)]}'
{"id":"openstack%2Fcinder~570885","triplet_id":"openstack%2Fcinder~master~I328a02d0f26e8c3d41ec18a0487da6fd20b39b04","project":"openstack/cinder","branch":"master","topic":"bug/1744692","hashtags":[],"change_id":"I328a02d0f26e8c3d41ec18a0487da6fd20b39b04","subject":"Add context to cloning snapshots in remotefs driver","status":"MERGED","created":"2018-05-29 10:19:37.000000000","updated":"2019-09-05 00:11:25.000000000","submitted":"2019-09-05 00:08:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"570885-1567642081613-05248e82","meta_rev_id":"88d6e671cc92215fbf23d835e534b51e32af46ef","_number":570885,"virtual_id_number":570885,"owner":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 09:37:36.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 11:01:07.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2019-07-10 15:31:30.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 10:29:50.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 20:37:01.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 12:08:30.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 12:05:52.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"date":"2019-07-10 15:10:38.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 22:07:55.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 13:02:34.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 09:10:38.000000000","_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":13628,"name":"Huawei Volume CI","username":"huawei-volume-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 10:45:40.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2019-07-10 11:20:28.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"value":0,"date":"2019-07-10 10:24:41.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":2,"date":"2019-09-05 00:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25677,"name":"Dell EMC PS CI","email":"dell.emc.ps.series.ci@emc.com","username":"dell.emc.ps.series.ci"},{"value":0,"date":"2019-07-10 12:52:37.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 12:32:00.000000000","_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2019-07-10 10:17:29.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2019-09-03 07:27:31.000000000","_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-07-10 11:29:35.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"date":"2019-07-10 10:30:37.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-07-10 12:09:53.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 08:42:16.000000000","_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 13:33:20.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2019-07-10 13:28:59.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 12:26:05.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 15:23:48.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-04 18:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-10 13:29:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":13628,"name":"Huawei Volume CI","username":"huawei-volume-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-17 08:01:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25677,"name":"Dell EMC PS CI","email":"dell.emc.ps.series.ci@emc.com","username":"dell.emc.ps.series.ci"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-04 22:06:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":13628,"name":"Huawei Volume CI","username":"huawei-volume-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25677,"name":"Dell EMC PS CI","email":"dell.emc.ps.series.ci@emc.com","username":"dell.emc.ps.series.ci"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":13628,"name":"Huawei Volume CI","username":"huawei-volume-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25677,"name":"Dell EMC PS CI","email":"dell.emc.ps.series.ci@emc.com","username":"dell.emc.ps.series.ci"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"_account_id":13628,"name":"Huawei Volume CI","username":"huawei-volume-ci","tags":["SERVICE_USER"]},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},{"_account_id":25677,"name":"Dell EMC PS CI","email":"dell.emc.ps.series.ci@emc.com","username":"dell.emc.ps.series.ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-29 23:36:09.000000000","updated_by":{"_account_id":13628,"name":"Huawei Volume CI","username":"huawei-volume-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13628,"name":"Huawei Volume CI","username":"huawei-volume-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-05-30 01:33:16.000000000","updated_by":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"reviewer":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"state":"REVIEWER"},{"updated":"2018-05-30 04:48:26.000000000","updated_by":{"_account_id":25677,"name":"Dell EMC PS CI","email":"dell.emc.ps.series.ci@emc.com","username":"dell.emc.ps.series.ci"},"reviewer":{"_account_id":25677,"name":"Dell EMC PS CI","email":"dell.emc.ps.series.ci@emc.com","username":"dell.emc.ps.series.ci"},"state":"REVIEWER"},{"updated":"2018-05-30 17:30:48.000000000","updated_by":{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-05-31 10:57:56.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-06-01 07:23:19.000000000","updated_by":{"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-01 09:24:13.000000000","updated_by":{"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-01 11:40:41.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-01-24 06:09:41.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2019-01-31 10:55:39.000000000","updated_by":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"reviewer":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"state":"REVIEWER"},{"updated":"2019-04-05 15:00:58.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-12 16:44:25.000000000","updated_by":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"reviewer":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"state":"REVIEWER"},{"updated":"2019-04-13 01:35:13.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2019-04-14 17:24:01.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-31 17:33:55.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2019-06-06 14:59:03.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 13:14:58.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-11 15:34:32.000000000","updated_by":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 11:17:36.000000000","updated_by":{"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},"reviewer":{"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},"state":"REVIEWER"},{"updated":"2019-06-12 13:34:15.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 15:49:55.000000000","updated_by":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 17:38:59.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-06-12 19:49:27.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 19:54:50.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-06-14 08:36:15.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-01 18:23:20.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-07-10 08:42:16.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 09:10:38.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-07-10 09:37:36.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 10:17:29.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-07-10 10:24:41.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 10:29:50.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 10:30:37.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-07-10 10:45:40.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 11:01:07.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-07-10 11:20:28.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 11:29:35.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 12:05:52.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 12:08:30.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2019-07-10 12:09:53.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 12:26:05.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 12:32:00.000000000","updated_by":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 12:52:37.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-07-10 13:02:34.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 13:28:59.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2019-07-10 13:29:08.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-07-10 13:33:20.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2019-07-10 15:10:38.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-07-10 15:23:48.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-07-10 15:31:30.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 20:37:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 22:07:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-17 08:01:38.000000000","updated_by":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"reviewer":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"state":"REVIEWER"},{"updated":"2019-09-04 22:06:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-05 00:08:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"269fccdd6b047585182a280296ffbc005ce1bc87","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-05-29 10:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b4b1799de3d01d34cfba69217b8e7e12d6ec622","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-05-29 10:23:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1811fa7aba5b5ad435c3cedb1a2553f974adc4d4","author":{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},"date":"2018-05-29 11:02:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- prophetstor-dsvm-tempest-cinder-ci http://download.prophetstor.com/prophetstor_ci/570885/1/check/prophetstor-dsvm-tempest-cinder-ci/12672 : SUCCESS in 41m 32s\n\nFor information of ProphetStor CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/ProphetStor_CI","accounts_in_message":[],"_revision_number":1},{"id":"03e55f1d0e910fa5e2eacbb444365f94ff5f86b3","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-05-29 11:24:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0df5dd68525a435a2d79da97e0b74e82180e8fd3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-05-29 11:26:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/1/check/tempest-dsvm-full-xenial/b1abb2d/ : SUCCESS in 1h 06m 28s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/1/check/devstack-plugin-ceph-dsvm-xenial/f98fa0d/ : SUCCESS in 1h 03m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"281ef99b6c2e215b8ea7259f81826f230de1eb2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-29 12:00:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/1/check/openstack-tox-pep8/d55ac85/ : SUCCESS in 8m 04s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/1/check/openstack-tox-py27/61ce5c7/ : FAILURE in 7m 33s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/1/check/openstack-tox-py35/acf677a/ : FAILURE in 7m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/570885/1/check/build-openstack-sphinx-docs/450d8c4/html/ : SUCCESS in 7m 03s\n- tempest-full http://logs.openstack.org/85/570885/1/check/tempest-full/81ccbb1/ : SUCCESS in 1h 29m 29s\n- neutron-grenade http://logs.openstack.org/85/570885/1/check/neutron-grenade/f238665/ : FAILURE in 4m 10s\n- tempest-full-py3 http://logs.openstack.org/85/570885/1/check/tempest-full-py3/2bc4b92/ : SUCCESS in 1h 35m 44s\n- openstack-tox-pylint http://logs.openstack.org/85/570885/1/check/openstack-tox-pylint/1c5bb6e/ : SUCCESS in 23m 58s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/570885/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/e4b4410/ : FAILURE in 4m 22s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/1/check/openstack-tox-functional/58af940/ : SUCCESS in 4m 02s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/1/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/35d4593/ : FAILURE in 7m 06s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/1/check/legacy-tempest-dsvm-lvm-multibackend/e6eb867/ : FAILURE in 5m 04s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/1/check/legacy-tempest-dsvm-full-drbd-devstack/c8b9198/ : FAILURE in 3m 10s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/1/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/f843860/ : FAILURE in 4m 32s (non-voting)\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/1/check/cinder-tempest-dsvm-lvm-lio-barbican/852a91e/ : FAILURE in 5m 04s\n- cinder-tox-compliance http://logs.openstack.org/85/570885/1/check/cinder-tox-compliance/e490ace/ : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/1/check/cinder-tox-bandit-baseline/06db155/ : SUCCESS in 6m 01s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/1/check/nova-multiattach/8c4bafb/ : FAILURE in 2m 51s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/1/check/openstack-tox-functional-py35/f68363a/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/1/check/openstack-tox-lower-constraints/436d004/ : FAILURE in 8m 06s\n- cinder-tox-functional-py36 http://logs.openstack.org/85/570885/1/check/cinder-tox-functional-py36/8867d0b/ : SUCCESS in 4m 33s\n- cinder-tox-py36 http://logs.openstack.org/85/570885/1/check/cinder-tox-py36/de5fc28/ : FAILURE in 7m 40s\n- cinder-rally-task http://logs.openstack.org/85/570885/1/check/cinder-rally-task/3a418ac/ : SUCCESS in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bb7f7aeeb5ede8faca7fd0bdc9b473b772f47a9","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-05-29 12:06:24.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/1/check/cinder-test-iscsi/26a9d20 : FAILURE in 19m 12s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/1/check/cinder-test-fc/23576a9 : FAILURE in 33m 17s\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/1/check/Cinder-AS13000/5b456ec : FAILURE in 22m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6f90a35e12f651008a850ddb1a7175c3d775a5a4","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-05-29 12:23:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/10726 : SUCCESS in 59m 20s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/7156 : FAILURE in 53s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/8950 : FAILURE in 34m 24s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/7545 : FAILURE in 39m 27s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/8516 : FAILURE in 1h 01m 46s\n- flashsystem-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-fc/7187 : FAILURE in 1h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"224b3a27ef54e3ca7081cf1b75c8fd65e8778ea2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-05-29 12:45:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/1/check-cinder/Cinder-ISER-ISCSI/4630e53 : SUCCESS in 59m 56s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/1/check-cinder/Cinder-ISER-LIO/596816b : SUCCESS in 59m 04s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"3ed00f68da99e3d5ce24555409a8725fa116b011","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-05-29 12:57:49.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/1/check/dsvm-veritas-access-cinder-driver/781fb4f : SUCCESS in 1h 22m 20s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/1/check/dsvm-veritas-access-cinder-iscsi-driver/347a026 : FAILURE in 1h 42m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a3e70ea8abe38aa8c68eba31607c74f61c200e55","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-05-29 13:34:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/1/check/dsvm-tempest-kvm/254d473 : FAILURE in 36m 48s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/1/check/dsvm-tempest-exe/f28020e : FAILURE in 47m 49s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"734dd04ab1396bcb7c8a0186e00fe43f4bb51989","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-05-29 13:53:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed590349c9e83f9c5edd2ef62a5fe45169623f3f","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-05-29 14:06:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b53ce0403a97fbb44f41a3e9f9c094b41cf9d3b","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-05-29 14:11:01.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nyeah, that\u0027s what I am mentioned in the launchpad :).","accounts_in_message":[],"_revision_number":2},{"id":"8059bb5129597bd1c2963740e2811d816409bf54","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-05-29 15:33:51.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/2/check/tempest-dsvm-full-xenial/e2d1588/ : SUCCESS in 1h 07m 03s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/2/check/devstack-plugin-ceph-dsvm-xenial/c806bc6/ : SUCCESS in 1h 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0079f4410c8eb664ad57d06325734d0fcfc03a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-29 15:51:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/2/check/openstack-tox-pep8/c3753c9/ : SUCCESS in 8m 56s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/2/check/openstack-tox-py27/847e339/ : SUCCESS in 8m 25s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/2/check/openstack-tox-py35/67bf79c/ : SUCCESS in 7m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/570885/2/check/build-openstack-sphinx-docs/c62dff9/html/ : SUCCESS in 7m 17s\n- tempest-full http://logs.openstack.org/85/570885/2/check/tempest-full/1f67e84/ : SUCCESS in 1h 23m 57s\n- neutron-grenade http://logs.openstack.org/85/570885/2/check/neutron-grenade/9a86581/ : SUCCESS in 1h 01m 23s\n- tempest-full-py3 http://logs.openstack.org/85/570885/2/check/tempest-full-py3/0157188/ : SUCCESS in 1h 11m 52s\n- openstack-tox-pylint http://logs.openstack.org/85/570885/2/check/openstack-tox-pylint/5a50ff5/ : SUCCESS in 26m 16s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/570885/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/acc1f50/ : SUCCESS in 1h 37m 59s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/2/check/openstack-tox-functional/ddf7c41/ : SUCCESS in 4m 45s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/2/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/42e3644/ : SUCCESS in 1h 02m 04s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/2/check/legacy-tempest-dsvm-lvm-multibackend/9b81783/ : SUCCESS in 1h 05m 45s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/2/check/legacy-tempest-dsvm-full-drbd-devstack/dc9392e/ : SUCCESS in 1h 34m 17s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/2/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/4753896/ : SUCCESS in 1h 13m 55s (non-voting)\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/2/check/cinder-tempest-dsvm-lvm-lio-barbican/b7e1391/ : SUCCESS in 1h 30m 03s\n- cinder-tox-compliance http://logs.openstack.org/85/570885/2/check/cinder-tox-compliance/3658ea8/ : SUCCESS in 4m 16s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/2/check/cinder-tox-bandit-baseline/743075d/ : SUCCESS in 6m 20s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/2/check/nova-multiattach/c0d85ee/ : SUCCESS in 48m 18s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/2/check/openstack-tox-functional-py35/f6b63c4/ : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/2/check/openstack-tox-lower-constraints/eba6405/ : SUCCESS in 8m 35s\n- cinder-tox-functional-py36 http://logs.openstack.org/85/570885/2/check/cinder-tox-functional-py36/cc880dd/ : SUCCESS in 4m 55s\n- cinder-tox-py36 http://logs.openstack.org/85/570885/2/check/cinder-tox-py36/f67990f/ : SUCCESS in 9m 17s\n- cinder-rally-task http://logs.openstack.org/85/570885/2/check/cinder-rally-task/758bd93/ : SUCCESS in 36m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5477c13ac326d2ff5613ad759d6fdf45a7a757e5","author":{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},"date":"2018-05-29 16:00:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- prophetstor-dsvm-tempest-cinder-ci http://download.prophetstor.com/prophetstor_ci/570885/2/check/prophetstor-dsvm-tempest-cinder-ci/12677 : SUCCESS in 50m 13s\n\nFor information of ProphetStor CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/ProphetStor_CI","accounts_in_message":[],"_revision_number":2},{"id":"7c492f3ce00c0dfe53ad028bc548ae7bf1664fb4","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-05-29 16:15:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/10735 : FAILURE in 50m 07s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/7165 : FAILURE in 44s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/8959 : FAILURE in 33m 33s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/7554 : FAILURE in 29s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/8525 : FAILURE in 1h 13m 28s\n- flashsystem-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-fc/7196 : FAILURE in 13s","accounts_in_message":[],"_revision_number":2},{"id":"f037cf83980973dca4975dc02a2548ad67d20f8e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-05-29 16:17:57.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/2/check-cinder/Cinder-ISER-ISCSI/d502bc9 : SUCCESS in 58m 57s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/2/check-cinder/Cinder-ISER-LIO/1db7ddd : FAILURE in 58m 19s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"83f96042870cca4d371b4e8c109d148ff65ee815","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-05-29 16:54:13.000000000","message":"Patch Set 2:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-85-570885-2-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a4b616702456dd78168f8c8cee77bce7c60e3cac","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-05-29 17:06:02.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"685fa1c352729d1a5940f4f753f13597ad6c12a3","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-05-29 17:10:07.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/2/check/cinder-test-iscsi/70bfb0b : FAILURE in 35m 45s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/2/check/cinder-test-fc/acde3ad : FAILURE in 35m 43s\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/2/check/Cinder-AS13000/0f5c5bf : FAILURE in 34m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a5c4b0e81742629c178460c8dc84fccb64f0d79b","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-05-29 17:20:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/2/check/dsvm-tempest-kvm/418b803 : FAILURE in 1h 02m 32s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/2/check/dsvm-tempest-exe/e7d083e : FAILURE in 44m 42s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"bd21e1a2cce5cd6eada76484ad655c1a1bb46486","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2018-05-29 17:32:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f4cd21acd3c37850847c14b8cc95af9aa174d80","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-05-29 17:55:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/85/570885/2/upstream-check/cinder-cDOT-iSCSI/92c4283/ : SUCCESS in 1h 16m 01s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/85/570885/2/upstream-check/cinder-cDOT-NFS/4c61f63/ : SUCCESS in 1h 21m 47s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/85/570885/2/upstream-check/cinder-cDOT-FCP/4783c58/ : FAILURE in 1h 23m 53s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8d9f786dca841424d5a67ae6bd4a1ab96ac67251","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-05-29 18:17:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/85/570885/2/upstream-eseries-check/cinder-eseries-iSCSI/ee1cf73/ : SUCCESS in 1h 53m 53s (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/85/570885/2/upstream-eseries-check/cinder-eseries-FCP/2decb7e/ : SUCCESS in 1h 07m 13s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b68b34bfdfd665370ba7a26c9d501246c976539","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-05-29 19:34:43.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/2/check/dsvm-veritas-access-cinder-driver/147e4b6 : SUCCESS in 1h 32m 45s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/2/check/dsvm-veritas-access-cinder-iscsi-driver/70896c3 : FAILURE in 2h 51m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a6c3373406940958f8def6ed573bafcf1247a2d8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-05-29 20:41:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/2/check/dsvm-tempest-infinibox-fc/1037410 : SUCCESS in 1h 13m 39s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/2/check/dsvm-tempest-infinibox-iscsi/1190979 : SUCCESS in 1h 20m 50s","accounts_in_message":[],"_revision_number":2},{"id":"61ab68150760e40a5817c030ba492d5a4b47f14a","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-05-29 21:29:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://172.22.1.21/85/570885/2/check/synology-dsm-cinder-iscsi/886575b : SUCCESS in 1h 36m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a8908b094a0870b191744aef913fd9225a70324b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-05-29 22:52:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/2/check/dsvm-tempest-storpool/afd3513 : SUCCESS in 1h 43m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4c180692b7248f44176494688aa555d8fbe7eb03","author":{"_account_id":13628,"name":"Huawei Volume CI","username":"huawei-volume-ci","tags":["SERVICE_USER"]},"date":"2018-05-29 23:36:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- huawei-fc-dsvm-tempest-full http://116.66.187.78:8088/huawei-fc-dsvm-tempest-full/5913 : SUCCESS in 39m 35s\n- huawei-iscsi-dsvm-tempest-full http://116.66.187.78:8088/huawei-iscsi-dsvm-tempest-full/5854 : SUCCESS in 1h 50m 37s","accounts_in_message":[],"_revision_number":2},{"id":"fc836f1f432d1d7871371870f5cca1afed4c16f2","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-05-30 00:52:52.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/85/570885/2/check/kaminario-dsvm-tempest-full-iscsi/69eed4d : FAILURE in 1h 46m 28s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/85/570885/2/check/kaminario-dsvm-tempest-full-FC/66dca02 : SUCCESS in 1h 40m 45s","accounts_in_message":[],"_revision_number":2},{"id":"deaa5c8ac48bc7d0822d90538bf8722a2611c3b5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-05-30 01:02:56.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- EMC_ScaleIO http://publiclogs.emc.com/570885/2/EMC_ScaleIO/844 : SUCCESS in 1h 25m 09s\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"d303dd1ef34c42eb2fb24338ee35bd488ac20238","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-05-30 01:33:16.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nThat\u0027s better:) https://github.com/openstack/oslo.versionedobjects/blob/337cd364590756190710ea45123f7afbd376e584/oslo_versionedobjects/base.py#L691","accounts_in_message":[],"_revision_number":2},{"id":"7163b55e8533c52f9c1e7c585581d4416284f9f9","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-05-30 02:50:53.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/2/check/NEC-ISCSI/53e2f5b : FAILURE in 58s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/2/check/NEC-FC/dc9baf0 : FAILURE in 1m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3082598a568d201154fa5aa6b4ea1eaae10b2687","author":{"_account_id":25677,"name":"Dell EMC PS CI","email":"dell.emc.ps.series.ci@emc.com","username":"dell.emc.ps.series.ci"},"date":"2018-05-30 04:48:26.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- Dell_PS_iSCSI http://publiclogs.emc.com/570885/2/Dell_PS_iSCSI/8 : SUCCESS in 2h 47m 55s\n\nLeave a comment with \u0027run-dell-emc-ps\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_PS_Series_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3c1819aa64d746ec4f6f5100bd09be4c0419a20","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-05-30 05:26:06.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder CI, add a review comment with run-Cloudbase Cinder CI\n\n- cinder-iscsi-full-tempest http://cloudbase-ci.com//cinder/570885/2/iscsi/ : FAILURE in 2h 13m 34s\n- cinder-smb3-windows-full-tempest http://cloudbase-ci.com//cinder/570885/2/smb3_windows/ : FAILURE in 3h 05m 13s","accounts_in_message":[],"_revision_number":2},{"id":"8f2dbcaff08a17e6648e6dacf3e640a5236ed8d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-05-30 10:35:31.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/2/check-not-vote/ext-cinder-zuul/fc5a088 : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":2},{"id":"332d751607b4586ef7133ba609a9b8832915aa47","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-05-30 16:13:09.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n@tommylike Yep, i initially couldn\u0027t place the launchpad proposal and when I was looking for alternatives in here I had totally forgotten about it, oops. :)","accounts_in_message":[],"_revision_number":2},{"id":"1de0e85d7c7f61efe36f3b82169c4ebecd9e7b50","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-05-30 16:45:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"473bde76434e0d36941e762a6f05e3042c3b5ae7","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-05-30 16:56:17.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/3/check/dsvm-veritas-access-cinder-driver/705024c : FAILURE in 3m 38s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/3/check/dsvm-veritas-access-cinder-iscsi-driver/d1a8fa1 : FAILURE in 1m 48s","accounts_in_message":[],"_revision_number":3},{"id":"4712c82cf1c64cd510ed8bc87a68a11df6ee7f69","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-05-30 16:59:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/3280415 : FAILURE in 9m 29s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e195a8e : FAILURE in 12m 57s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"8cbbc58057d337c7a8b4eb98b938b3ba66c6e738","author":{"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},"date":"2018-05-30 17:27:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- cinder-dsvm-tempest-cisco-zonemanager-current-ubuntu-czm http://192.133.156.18/85/570885/3/check/cinder-dsvm-tempest-cisco-zonemanager-current-ubuntu-czm/6d369b7/ : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":3},{"id":"2b840d9cd2fc890f1434e86a879b33c189927419","author":{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},"date":"2018-05-30 17:30:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- prophetstor-dsvm-tempest-cinder-ci http://download.prophetstor.com/prophetstor_ci/570885/3/check/prophetstor-dsvm-tempest-cinder-ci/12694 : SUCCESS in 40m 24s\n\nFor information of ProphetStor CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/ProphetStor_CI","accounts_in_message":[],"_revision_number":3},{"id":"40696389d410e35901bf181f528919bebaeb35bc","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-05-30 17:44:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/3/check/dsvm-tempest-kvm/af7d87b : FAILURE in 57m 37s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/3/check/dsvm-tempest-exe/5d861c0 : FAILURE in 46m 20s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"f0583785719493895aa7c3a1df9abc41c89c6807","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-05-30 17:52:58.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/3/check/tempest-dsvm-full-xenial/5b24013/ : SUCCESS in 1h 04m 13s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/3/check/devstack-plugin-ceph-dsvm-xenial/2d75e8d/ : SUCCESS in 1h 06m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"00ebe5420cb1855e84e3e273be2b26dc63243f6c","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-05-30 17:53:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/10759 : SUCCESS in 56m 06s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/7189 : FAILURE in 58s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/8983 : SUCCESS in 56m 20s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/7579 : FAILURE in 54s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/8549 : FAILURE in 1h 07m 04s\n- flashsystem-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-fc/7220 : FAILURE in 33s","accounts_in_message":[],"_revision_number":3},{"id":"de923079aced83da2a4169dca513d25f657589b4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-05-30 18:15:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/3/check-cinder/Cinder-ISER-ISCSI/82f007c : SUCCESS in 57m 48s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/3/check-cinder/Cinder-ISER-LIO/a7e95fe : SUCCESS in 1h 03m 03s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"24dbdd88b96a63ce9db92d4cd8c015d538bbc0f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-30 18:32:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/3/check/openstack-tox-pep8/ab9989a/ : SUCCESS in 8m 32s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/3/check/openstack-tox-py27/202be47/ : SUCCESS in 8m 23s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/3/check/openstack-tox-py35/8ff97e6/ : SUCCESS in 8m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/570885/3/check/build-openstack-sphinx-docs/62e63a9/html/ : SUCCESS in 8m 00s\n- tempest-full http://logs.openstack.org/85/570885/3/check/tempest-full/395b93e/ : SUCCESS in 1h 26m 45s\n- neutron-grenade http://logs.openstack.org/85/570885/3/check/neutron-grenade/f775ca4/ : SUCCESS in 48m 35s\n- tempest-full-py3 http://logs.openstack.org/85/570885/3/check/tempest-full-py3/1ed39f6/ : FAILURE in 1h 03m 28s\n- openstack-tox-pylint http://logs.openstack.org/85/570885/3/check/openstack-tox-pylint/88b6ac0/ : SUCCESS in 23m 24s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/04a131a/ : SUCCESS in 1h 17m 02s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/3/check/openstack-tox-functional/75b2613/ : SUCCESS in 4m 40s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/3/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/861d905/ : SUCCESS in 54m 26s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-lvm-multibackend/ff45631/ : SUCCESS in 1h 12m 26s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-drbd-devstack/28ba294/ : SUCCESS in 1h 14m 22s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/16e943b/ : SUCCESS in 1h 38m 43s (non-voting)\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/3/check/cinder-tempest-dsvm-lvm-lio-barbican/6949411/ : FAILURE in 1h 37m 58s\n- cinder-tox-compliance http://logs.openstack.org/85/570885/3/check/cinder-tox-compliance/3214a3b/ : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/3/check/cinder-tox-bandit-baseline/41ba630/ : SUCCESS in 6m 26s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/3/check/nova-multiattach/bc5668f/ : SUCCESS in 46m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/3/check/openstack-tox-functional-py35/c7b74b0/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/3/check/openstack-tox-lower-constraints/fa463ea/ : SUCCESS in 9m 08s\n- cinder-tox-functional-py36 http://logs.openstack.org/85/570885/3/check/cinder-tox-functional-py36/296a121/ : SUCCESS in 5m 53s\n- cinder-tox-py36 http://logs.openstack.org/85/570885/3/check/cinder-tox-py36/9c11ef5/ : SUCCESS in 8m 36s\n- cinder-rally-task http://logs.openstack.org/85/570885/3/check/cinder-rally-task/a26e74b/ : SUCCESS in 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ea146dbca4e82ccc00b33b85b8d6005a11d6a3ae","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-05-30 18:32:27.000000000","message":"Patch Set 3:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- cinder-tempest-dsvm-lvm-lio-barbican: https://bugs.launchpad.net/bugs/1770179\n- tempest-full-py3: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"526b8603c56ac589a4da5dcf3f74c5aa10406e54","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-05-30 18:44:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/85/570885/3/check/dsvm-tempest-zadara-driver/97cf426 : SUCCESS in 1h 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5665750c0abbc45ce402e7f4afe7e3c351ea6c86","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-05-30 18:51:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/3/check/dsvm-tempest-infinibox-fc/ddb3198 : SUCCESS in 1h 15m 10s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/3/check/dsvm-tempest-infinibox-iscsi/083931c : SUCCESS in 1h 25m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c7200366f1f7a42f92cdb80297862ed1fc6e652b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-05-30 19:38:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/3/check/dsvm-tempest-storpool/4bec606 : SUCCESS in 1h 46m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f226ed2115f70b5d50ea7450032f15800244f8fd","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-05-30 20:12:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/3/check/NEC-ISCSI/0b7485f : SUCCESS in 1h 50m 44s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/3/check/NEC-FC/ddbfc4a : SUCCESS in 1h 56m 30s","accounts_in_message":[],"_revision_number":3},{"id":"db94e6684e680a2ab727402475d2f1506aaecc4b","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-05-30 22:37:20.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/3/check/cinder-test-iscsi/c5bed9d : SUCCESS in 1h 14m 11s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/3/check/cinder-test-fc/e45ed69 : FAILURE in 2h 31m 11s\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/3/check/Cinder-AS13000/909dead : SUCCESS in 2h 22m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ba19a1040e13336e52697e59dbab34e4385db10f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-05-30 23:01:14.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/3/check-not-vote/ext-cinder-zuul/3d1f20c : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":3},{"id":"343778c35617fd6f21d4bfe0ca95f1741683e6df","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-05-31 09:14:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"48ea94d143f0c0ec87dbd436cca848b5f278ea58","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-05-31 09:51:06.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/3/check-not-vote/ext-cinder-zuul/12b59aa : FAILURE in 36m 10s","accounts_in_message":[],"_revision_number":3},{"id":"d769b8502191a4897d9afa7966a124299a8d9c93","author":{"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},"date":"2018-05-31 10:03:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- cinder-dsvm-tempest-cisco-zonemanager-current-ubuntu-czm http://192.133.156.18/85/570885/3/check/cinder-dsvm-tempest-cisco-zonemanager-current-ubuntu-czm/ec45178/ : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":3},{"id":"1be1cb093b6821df69b2d6f41d57360d25718729","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-05-31 10:05:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/b4e1233 : SUCCESS in 49m 52s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/be76ed6 : SUCCESS in 50m 23s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"1e38fa03a93ad26a8be4b17c8658ddc0e461431d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-05-31 10:15:05.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/3/check/tempest-dsvm-full-xenial/6e8b8c5/ : FAILURE in 1h 00m 17s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/3/check/devstack-plugin-ceph-dsvm-xenial/b4380b4/ : FAILURE in 53m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d3d835e6b89b2451025ea14baf7d97965752617","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-05-31 10:19:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/3/check/dsvm-tempest-kvm/2d3cd1a : FAILURE in 59m 52s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/3/check/dsvm-tempest-exe/5490e38 : FAILURE in 48m 54s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"24abb0e931bbe7fced3c92d79ec82e3f7692c552","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-31 10:57:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/3/check/openstack-tox-pep8/041030e/ : SUCCESS in 7m 57s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/3/check/openstack-tox-py27/f13bcf0/ : SUCCESS in 7m 31s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/3/check/openstack-tox-py35/fe11165/ : SUCCESS in 7m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/570885/3/check/build-openstack-sphinx-docs/afea20e/html/ : SUCCESS in 6m 44s\n- tempest-full http://logs.openstack.org/85/570885/3/check/tempest-full/2482bb9/ : SUCCESS in 1h 37m 01s\n- neutron-grenade http://logs.openstack.org/85/570885/3/check/neutron-grenade/d758f82/ : SUCCESS in 54m 21s\n- tempest-full-py3 http://logs.openstack.org/85/570885/3/check/tempest-full-py3/57ae01d/ : SUCCESS in 1h 13m 21s\n- openstack-tox-pylint http://logs.openstack.org/85/570885/3/check/openstack-tox-pylint/7249962/ : SUCCESS in 24m 58s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/76a2c2b/ : SUCCESS in 1h 11m 42s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/3/check/openstack-tox-functional/c49ca38/ : SUCCESS in 3m 42s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/3/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/83f8f7e/ : SUCCESS in 52m 39s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-lvm-multibackend/0fe02eb/ : SUCCESS in 1h 33m 58s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-drbd-devstack/deeae19/ : SUCCESS in 1h 15m 19s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/29f4dc2/ : SUCCESS in 1h 24m 41s (non-voting)\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/3/check/cinder-tempest-dsvm-lvm-lio-barbican/569d2c0/ : FAILURE in 1h 31m 11s\n- cinder-tox-compliance http://logs.openstack.org/85/570885/3/check/cinder-tox-compliance/a9ba865/ : SUCCESS in 4m 12s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/3/check/cinder-tox-bandit-baseline/5f0bfb2/ : SUCCESS in 7m 23s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/3/check/nova-multiattach/add022b/ : SUCCESS in 51m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/3/check/openstack-tox-functional-py35/ca7c5cd/ : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/3/check/openstack-tox-lower-constraints/4fe5ce3/ : SUCCESS in 8m 15s\n- cinder-tox-functional-py36 http://logs.openstack.org/85/570885/3/check/cinder-tox-functional-py36/ac5e93c/ : SUCCESS in 6m 34s\n- cinder-tox-py36 http://logs.openstack.org/85/570885/3/check/cinder-tox-py36/9e17724/ : SUCCESS in 8m 09s\n- cinder-rally-task http://logs.openstack.org/85/570885/3/check/cinder-rally-task/73db1b5/ : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1564a77aa2aa45b187bfa9ac0bab3e9816b01d50","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-05-31 10:57:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/10769 : SUCCESS in 52m 07s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/7199 : FAILURE in 19s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/8993 : SUCCESS in 54m 11s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/7589 : FAILURE in 1h 39m 07s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/8559 : SUCCESS in 1h 17m 50s\n- flashsystem-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-fc/7230 : FAILURE in 25s","accounts_in_message":[],"_revision_number":3},{"id":"3894acc473716e38a8f5631374a763a7ee1bcd7b","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-05-31 10:57:56.000000000","message":"Patch Set 3:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- cinder-tempest-dsvm-lvm-lio-barbican: https://bugs.launchpad.net/bugs/1763712\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"f569eccf330664b00de5521ccc2781ee7429b6a3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-05-31 11:36:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/85/570885/3/check/dsvm-tempest-zadara-driver/26a94a4 : SUCCESS in 1h 22m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eb8fd3659ba5cd7ea8645f4ef83f6e4d82cdc708","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-05-31 12:29:12.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder CI, add a review comment with run-Cloudbase Cinder CI\n\n- cinder-iscsi-full-tempest http://cloudbase-ci.com//cinder/570885/3/iscsi/ : SUCCESS in 1h 01m 37s\n- cinder-smb3-windows-full-tempest http://cloudbase-ci.com//cinder/570885/3/smb3_windows/ : FAILURE in 2h 32m 39s","accounts_in_message":[],"_revision_number":3},{"id":"dd3d2d0a3defaa3ee179b7e75f1fbddaff5dad7a","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-06-01 06:24:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"da5e69ded33763b019362d1a101fd251b82b617c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-01 07:08:50.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/3/check-not-vote/ext-cinder-zuul/5765419 : FAILURE in 43m 35s","accounts_in_message":[],"_revision_number":3},{"id":"c12d18d030a83cc6772d3cd2ab52e2bfdf7769a1","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-06-01 07:13:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c074de4 : SUCCESS in 48m 47s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/39d29b1 : SUCCESS in 45m 28s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"82f6ef183f7e59d410064fb372b982eef43e3278","author":{"_account_id":14208,"name":"Cisco CI","email":"ml2.ci@cisco.com","username":"cisco_ml2_ci","tags":["SERVICE_USER"]},"date":"2018-06-01 07:23:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- cinder-dsvm-tempest-cisco-zonemanager-current-ubuntu-czm http://192.133.156.18/85/570885/3/check/cinder-dsvm-tempest-cisco-zonemanager-current-ubuntu-czm/e657297/ : SUCCESS in 38m 18s","accounts_in_message":[],"_revision_number":3},{"id":"c2253a39e5395f0b90aa351c244beb4ebc817944","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-06-01 07:31:03.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/3/check/tempest-dsvm-full-xenial/452350d/ : SUCCESS in 1h 05m 55s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/3/check/devstack-plugin-ceph-dsvm-xenial/1ed3b3d/ : SUCCESS in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f3140ecc7517443dcc4cca9f3cc9a3fbfd3a6e74","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-06-01 07:44:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/10792 : FAILURE in 49m 45s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/7222 : FAILURE in 41s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/9016 : SUCCESS in 56m 34s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/7612 : FAILURE in 49s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/8582 : FAILURE in 1h 19m 54s\n- flashsystem-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-fc/7253 : FAILURE in 27s","accounts_in_message":[],"_revision_number":3},{"id":"4a4949f34be0eb5388b99f392ecbed2a0e10a303","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-01 08:07:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/3/check/openstack-tox-pep8/c8b1ab9/ : SUCCESS in 8m 41s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/3/check/openstack-tox-py27/0f58da7/ : SUCCESS in 9m 15s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/3/check/openstack-tox-py35/b6be4c6/ : SUCCESS in 8m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/570885/3/check/build-openstack-sphinx-docs/542695d/html/ : SUCCESS in 7m 25s\n- tempest-full http://logs.openstack.org/85/570885/3/check/tempest-full/aaa887d/ : SUCCESS in 1h 25m 16s\n- neutron-grenade http://logs.openstack.org/85/570885/3/check/neutron-grenade/4a0a594/ : SUCCESS in 46m 30s\n- tempest-full-py3 http://logs.openstack.org/85/570885/3/check/tempest-full-py3/9f9d72b/ : SUCCESS in 1h 21m 04s\n- openstack-tox-pylint http://logs.openstack.org/85/570885/3/check/openstack-tox-pylint/c69cdb2/ : SUCCESS in 28m 46s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/6a46444/ : SUCCESS in 1h 09m 35s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/3/check/openstack-tox-functional/cd5d087/ : SUCCESS in 5m 08s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/3/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/136dc64/ : SUCCESS in 55m 04s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-lvm-multibackend/bc59667/ : SUCCESS in 1h 22m 02s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-drbd-devstack/5f5c77f/ : SUCCESS in 1h 06m 39s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/3/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/c4c132f/ : SUCCESS in 1h 27m 58s (non-voting)\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/3/check/cinder-tempest-dsvm-lvm-lio-barbican/c2f2006/ : SUCCESS in 1h 36m 37s\n- cinder-tox-compliance http://logs.openstack.org/85/570885/3/check/cinder-tox-compliance/ca0d0bc/ : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/3/check/cinder-tox-bandit-baseline/77c3550/ : SUCCESS in 6m 45s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/3/check/nova-multiattach/b9ee59d/ : SUCCESS in 1h 05m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/3/check/openstack-tox-functional-py35/3b5d233/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/3/check/openstack-tox-lower-constraints/39900a4/ : SUCCESS in 8m 00s\n- cinder-tox-functional-py36 http://logs.openstack.org/85/570885/3/check/cinder-tox-functional-py36/ba242b1/ : SUCCESS in 5m 58s\n- cinder-tox-py36 http://logs.openstack.org/85/570885/3/check/cinder-tox-py36/fcc808d/ : SUCCESS in 9m 05s\n- cinder-rally-task http://logs.openstack.org/85/570885/3/check/cinder-rally-task/cb30ba8/ : SUCCESS in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"feb416a00e5c3c0466ddc1aad33a936233d74245","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-06-01 08:40:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/3/check/dsvm-tempest-kvm/2e90e40 : FAILURE in 1h 24m 27s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/3/check/dsvm-tempest-exe/0826e8e : FAILURE in 56m 09s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"cceef1e78582c0ae719a38b5eff956ed080fb798","author":{"_account_id":10622,"name":"SolidFire CI","email":"openstack.ci@solidfire.com","username":"sfci","tags":["SERVICE_USER"]},"date":"2018-06-01 09:24:13.000000000","message":"Patch Set 3:\n\n* SolidFire-dsvm-full http://54.164.167.86/solidfire-ci-logs/refs-changes-85-570885-3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"421b81bd3872eec3bc20de276c4c4f0a7c097f91","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-06-01 09:41:54.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"4261f28ede1e72ce63264790c68b68c1df2d9b8e","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-06-01 11:40:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/85/570885/3/upstream-eseries-check/cinder-eseries-iSCSI/24159d7/ : SUCCESS in 3h 08m 00s (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/85/570885/3/upstream-eseries-check/cinder-eseries-FCP/6a504af/ : SUCCESS in 2h 14m 33s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":3},{"id":"a5e153e5185d4b3380078a074fbac4c5688d3763","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-06-01 12:38:27.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"4c3c7c81c0758e51203c89694e7ccee108e911bb","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-06-01 12:39:09.000000000","message":"Patch Set 3:\n\nI manually tested all discussed variants. In all cases the context attribute is missing, this needs some further thought.","accounts_in_message":[],"_revision_number":3},{"id":"63f2ed1b36133611c7381ffcf9ae67bd55f821c2","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-06-01 13:17:05.000000000","message":"Patch Set 3:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-85-570885-3-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a080ec3dde8c520e965c47fc8b408f34daaeebeb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-06-01 14:41:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/85/570885/3/upstream-check/cinder-cDOT-iSCSI/1b87604/ : SUCCESS in 3h 02m 03s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/85/570885/3/upstream-check/cinder-cDOT-NFS/a7b26cf/ : SUCCESS in 3h 07m 06s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/85/570885/3/upstream-check/cinder-cDOT-FCP/ad0951e/ : FAILURE in 2h 21m 12s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0808edaf95e1f5581ebb320ab30218019a6a2f8e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-06-01 16:12:45.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder CI, add a review comment with run-Cloudbase Cinder CI\n\n- cinder-iscsi-full-tempest http://cloudbase-ci.com//cinder/570885/3/iscsi/ : SUCCESS in 1h 02m 18s\n- cinder-smb3-windows-full-tempest http://cloudbase-ci.com//cinder/570885/3/smb3_windows/ : FAILURE in 2h 36m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c44d37e6b840534cf7bb66316b2b97b1e5631c37","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-06-01 16:53:35.000000000","message":"Patch Set 3:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/85/570885/3/HPMSA-FC : SUCCESS in 27:25.11\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/85/570885/3/HPMSA-iSCSI : SUCCESS in 32:31.23","accounts_in_message":[],"_revision_number":3},{"id":"c7fcef1cd7641c78d4bc8eb72aa7207e34295ddb","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-06-01 17:41:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/85/570885/3/check/dsvm-tempest-zadara-driver/cd554f3 : SUCCESS in 1h 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"683d764a9a68b7a7ed35ac55ff13734d82a85e37","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-06-01 19:59:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://172.22.1.21/85/570885/3/check/synology-dsm-cinder-iscsi/8bfbcd0 : SUCCESS in 1h 36m 31s","accounts_in_message":[],"_revision_number":3},{"id":"50d962627801054bf970e123311a7bc99812112f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-06-01 22:56:50.000000000","message":"Patch Set 3:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/85/570885/3/Lenovo-FC : SUCCESS in 26:31.52\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/85/570885/3/Lenovo-iSCSI : SUCCESS in 31:48.83","accounts_in_message":[],"_revision_number":3},{"id":"74f9593c63827a908c3fd70993f1d95d4e8e3894","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-06-02 02:30:55.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/85/570885/3/check/kaminario-dsvm-tempest-full-iscsi/232e051 : SUCCESS in 1h 40m 56s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/85/570885/3/check/kaminario-dsvm-tempest-full-FC/d2f8ee9 : FAILURE in 1h 37m 30s","accounts_in_message":[],"_revision_number":3},{"id":"f6c8f57661437f582fd50b2b95981b702590ff17","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-06-05 03:30:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/570885/3/EMC_Unity_FC/441 : SUCCESS in 2h 00m 53s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/570885/3/EMC_Unity_iSCSI/418 : FAILURE in 2h 47m 59s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4d026e53218e028266c5b09a238cb02396c7d483","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2018-07-10 13:52:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d9059cbcf59870b28e673f99bcbfcc92e019c5f3","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-12-19 11:27:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cee312fb99d805e9a2e2dfe9f712687a1d6fa087","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-19 11:58:51.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/429 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/838 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/669 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"1d7a9071aa45ed7cd7c1e7bd01ff093fb28165eb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-19 12:07:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/4/check/dsvm-tempest-cisco-zonemanager-job/d49b286/ : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":4},{"id":"5c3e36616983abea58f46a1782a2480ad69119bc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-19 12:20:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/4/iscsi/ : SUCCESS in 52m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/4/smb/ : SUCCESS in 48m 20s","accounts_in_message":[],"_revision_number":4},{"id":"5c0a1440e7d6bce9ffed11a87d5b72ea61747acb","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-19 13:10:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/4/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/e659e72 : SUCCESS in 48m 19s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/4/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/2cb8ef7 : FAILURE in 1h 42m 58s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"37e3fae846baaedf02def36131362e67a32ccd16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-19 13:33:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/4/check/openstack-tox-lower-constraints/06af4a1/ : FAILURE in 9m 39s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/4/check/openstack-tox-pep8/60af04e/ : FAILURE in 7m 10s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/4/check/openstack-tox-py27/df56105/ : FAILURE in 9m 31s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/4/check/openstack-tox-py35/4db6ef6/ : FAILURE in 8m 38s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/4/check/openstack-tox-py36/5d0dfab/ : FAILURE in 8m 03s\n- openstack-tox-docs http://logs.openstack.org/85/570885/4/check/openstack-tox-docs/19c355f/html/ : SUCCESS in 7m 23s\n- tempest-full http://logs.openstack.org/85/570885/4/check/tempest-full/d4bcc14/ : SUCCESS in 1h 19m 01s\n- neutron-grenade http://logs.openstack.org/85/570885/4/check/neutron-grenade/293c47b/ : SUCCESS in 54m 17s\n- grenade-py3 http://logs.openstack.org/85/570885/4/check/grenade-py3/6ab1eac/ : SUCCESS in 51m 06s\n- tempest-full-py3 http://logs.openstack.org/85/570885/4/check/tempest-full-py3/2909904/ : SUCCESS in 1h 34m 36s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/4/check/cinder-tempest-dsvm-lvm-lio-barbican/bac6c9a/ : SUCCESS in 1h 45m 42s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/4/check/cinder-tox-bandit-baseline/a42bee5/ : SUCCESS in 5m 52s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/4/check/nova-multiattach/91fc394/ : SUCCESS in 1h 13m 10s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/4/check/openstack-tox-functional-py35/62d7ee1/ : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/4/check/openstack-tox-functional-py36/415b9a2/ : SUCCESS in 4m 22s\n- cinder-rally-task http://logs.openstack.org/85/570885/4/check/cinder-rally-task/e298712/ : SUCCESS in 32m 24s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/4/check/openstack-tox-pylint/9853d5c/ : FAILURE in 4m 21s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/4/check/devstack-plugin-ceph-tempest/b457f8d/ : SUCCESS in 1h 51m 01s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/4/check/openstack-tox-functional/b380712/ : SUCCESS in 4m 34s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/4/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/312f285/ : SUCCESS in 1h 09m 24s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/4/check/legacy-tempest-dsvm-lvm-multibackend/069917e/ : SUCCESS in 1h 47m 14s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/4/check/legacy-tempest-dsvm-full-drbd-devstack/fbc1e37/ : SUCCESS in 1h 08m 51s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/4/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/8d1facd/ : FAILURE in 1h 13m 06s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/4/check/tempest-slow/fcaf7cd/ : SUCCESS in 2h 03m 31s","accounts_in_message":[],"_revision_number":4},{"id":"b08e8c3ac6320b42caea57065e8ac2fe8d9b63f9","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-19 14:03:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/4/check/dsvm-tempest-kvm/20a9461 : SUCCESS in 1h 10m 48s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/4/check/dsvm-tempest-exe/a3e5df3 : SUCCESS in 57m 55s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"91477ee85667ddd8cd3e419a221beeefbc3b234b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-19 14:06:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/4/check-cinder/Cinder-ISER-ISCSI/3ef2836 : SUCCESS in 54m 11s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/4/check-cinder/Cinder-ISER-LIO/ea2c584 : SUCCESS in 58m 35s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"c7927b47cc36d2d39d24fe966d9dfcdc5dac4525","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-19 14:30:10.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/4/check/tempest-dsvm-full-xenial/2ec2931/ : FAILURE in 3h 02m 20s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/4/check/devstack-plugin-ceph-dsvm-xenial/e452a43/ : FAILURE in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5a8dd16d7b3f04987b80b974a31b51d29d6849cb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-19 14:42:42.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/4/check-not-vote/ext-cinder-zuul/2959186 : FAILURE in 49m 09s","accounts_in_message":[],"_revision_number":4},{"id":"5fdc48a5d18e860e1f725c5845275c75167ca2e1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-19 15:37:38.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/4/check/fujitsu-eternus-dx-iscsi/5541d03 : SUCCESS in 2h 15m 47s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/4/check/fujitsu-eternus-dx-fc/69d2899 : SUCCESS in 2h 06m 40s","accounts_in_message":[],"_revision_number":4},{"id":"86f1113c4725d6be0ecb498d9ac35cfac3745af1","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-12-19 16:41:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"65eda3858e0569f241b0fcd1de3e526350e4d63f","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-12-19 16:46:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"af083e76d5e1db4a8281012bd3dee7da3070ec0f","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-19 16:59:04.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/437 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/841 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/672 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"a5bbb23563b3699d363d9c58cb13b09a57f41402","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-19 17:15:16.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/438 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/842 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/673 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"75a97b7ca90d9d0faef3621290aa5ca26855234f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-19 18:10:01.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/6/iscsi/ : FAILURE in 1h 23m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/6/smb/ : SUCCESS in 48m 39s","accounts_in_message":[],"_revision_number":6},{"id":"619d7dde9ae9405d792bad2946b3f6658531c7df","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-19 18:16:10.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/6/check/tempest-dsvm-full-xenial/5c2a80c/ : SUCCESS in 1h 09m 10s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/6/check/devstack-plugin-ceph-dsvm-xenial/6ea52f6/ : SUCCESS in 1h 09m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"00031cab93935511cc80476d85cfb3ba7ad0401c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-19 19:36:47.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/6/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/d303c1b : FAILURE in 51m 10s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/6/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/f939b60 : FAILURE in 1m 27s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"2d9964a15be5f53f52674e600530a56acc97396d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-19 20:01:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/6/check/openstack-tox-lower-constraints/c50e10b/ : SUCCESS in 10m 05s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/6/check/openstack-tox-pep8/d2a9402/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/6/check/openstack-tox-py27/623a6d7/ : SUCCESS in 8m 09s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/6/check/openstack-tox-py35/db43422/ : SUCCESS in 8m 09s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/6/check/openstack-tox-py36/495c42b/ : SUCCESS in 7m 45s\n- openstack-tox-docs http://logs.openstack.org/85/570885/6/check/openstack-tox-docs/c673850/html/ : SUCCESS in 6m 46s\n- tempest-full http://logs.openstack.org/85/570885/6/check/tempest-full/bab5e34/ : SUCCESS in 1h 19m 15s\n- neutron-grenade http://logs.openstack.org/85/570885/6/check/neutron-grenade/e2a7762/ : SUCCESS in 49m 45s\n- grenade-py3 http://logs.openstack.org/85/570885/6/check/grenade-py3/017af3b/ : SUCCESS in 56m 59s\n- tempest-full-py3 http://logs.openstack.org/85/570885/6/check/tempest-full-py3/0ec84b9/ : SUCCESS in 1h 21m 50s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/6/check/cinder-tempest-dsvm-lvm-lio-barbican/9629955/ : SUCCESS in 2h 25m 50s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/6/check/cinder-tox-bandit-baseline/ac6f7fd/ : SUCCESS in 6m 05s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/6/check/nova-multiattach/6c0288f/ : SUCCESS in 53m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/6/check/openstack-tox-functional-py35/8532088/ : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/6/check/openstack-tox-functional-py36/60f7e82/ : SUCCESS in 6m 42s\n- cinder-rally-task http://logs.openstack.org/85/570885/6/check/cinder-rally-task/ffa06ce/ : SUCCESS in 41m 29s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/6/check/openstack-tox-pylint/b972a91/ : FAILURE in 4m 18s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/6/check/devstack-plugin-ceph-tempest/3d5b338/ : SUCCESS in 1h 33m 34s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/6/check/openstack-tox-functional/43c1feb/ : SUCCESS in 5m 19s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/6/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/a419606/ : SUCCESS in 55m 41s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/6/check/legacy-tempest-dsvm-lvm-multibackend/9edd4e0/ : FAILURE in 1h 13m 19s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/6/check/legacy-tempest-dsvm-full-drbd-devstack/ea2c241/ : SUCCESS in 1h 16m 08s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/6/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/8ddc748/ : FAILURE in 1h 00m 18s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/6/check/tempest-slow/a269b8b/ : SUCCESS in 1h 37m 12s","accounts_in_message":[],"_revision_number":6},{"id":"cac0a61d3903fb5db1dee031ffd504fb5026a477","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-19 20:07:36.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/6/check/fujitsu-eternus-dx-iscsi/f0b1b18 : SUCCESS in 2h 14m 08s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/6/check/fujitsu-eternus-dx-fc/843e27c : SUCCESS in 2h 04m 45s","accounts_in_message":[],"_revision_number":6},{"id":"0f2a8c7d6274604d71de381bbd2fa26b853ff0b8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-19 21:38:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/6/check/dsvm-tempest-kvm/8af3831 : SUCCESS in 1h 01m 53s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/6/check/dsvm-tempest-exe/e7a93b8 : FAILURE in 44m 36s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"a6598c757deca559232bc2429ec4856524b39b99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-19 21:42:33.000000000","message":"Patch Set 6:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"e0dac51b006348e8946c5e41b209eac0657dc0e7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-19 22:19:42.000000000","message":"Patch Set 6:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/676cf52 : SUCCESS in 1h 57m 11s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/6/check/3par-fc-driver-master-client-pip-ssa02-dsvm/4425998 : FAILURE in 1h 57m 15s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/85/570885/6/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/8322b8b : SUCCESS in 1h 50m 32s","accounts_in_message":[],"_revision_number":6},{"id":"aaebfe0adad3f7992d937195c6b88a2eb6ba4749","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-19 23:22:03.000000000","message":"Patch Set 6:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/6/check/dsvm-veritas-access-cinder-driver/c4276ee : FAILURE in 1h 32m 04s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/6/check/dsvm-veritas-access-cinder-iscsi-driver/df606da : FAILURE in 11s","accounts_in_message":[],"_revision_number":6},{"id":"1754c8e613f7e99fd35b38940ca003843bf0c77c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-19 23:40:02.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/6/check-not-vote/ext-cinder-zuul/b197d3d : FAILURE in 26m 43s","accounts_in_message":[],"_revision_number":6},{"id":"6ffd191675bc014855822f2a2ea0d514c2d27615","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-19 23:48:26.000000000","message":"Patch Set 6:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/85/570885/6/check/kaminario-dsvm-tempest-full-iscsi/d50208e : FAILURE in 1m 26s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/85/570885/6/check/kaminario-dsvm-tempest-full-FC/15d9048 : FAILURE in 1m 26s","accounts_in_message":[],"_revision_number":6},{"id":"8dffe6633ba75c59f1be5ab042c95c2ea6ece030","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-20 01:42:36.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/6/check/NEC-ISCSI/6e2bd5e : FAILURE in 2s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/6/check/NEC-FC/8cdbf08 : FAILURE in 2s","accounts_in_message":[],"_revision_number":6},{"id":"8ec4685bdd70f5f0dc226e00728b0bf8336f5270","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-20 07:19:41.000000000","message":"Patch Set 6:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/85/570885/6/HPMSA-FC : SUCCESS in 37:06.80\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/85/570885/6/HPMSA-iSCSI : SUCCESS in 41:13.68","accounts_in_message":[],"_revision_number":6},{"id":"01e3c354bb201430f966de492de3c6428c8abc45","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-20 09:47:17.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/6/check/dsvm-tempest-storpool/afb376e : SUCCESS in 1h 42m 06s","accounts_in_message":[],"_revision_number":6},{"id":"de2b1b6f9bc649083159b5fae1387153ddc4f9bb","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-12-20 10:22:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"896c7fd2d460c5ac4ee32c252169deaa8c433fbe","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-20 10:23:05.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/7/check/NEC-ISCSI/d5917bd : FAILURE in 8s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/7/check/NEC-FC/4c3d6b6 : FAILURE in 2s","accounts_in_message":[],"_revision_number":7},{"id":"02ee12ea53b2e566fadee830c15ad6f03a1aae44","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-20 10:23:21.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/7/check/dsvm-veritas-access-cinder-driver/0f319b3 : FAILURE in 12s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/7/check/dsvm-veritas-access-cinder-iscsi-driver/be01bc9 : FAILURE in 12s","accounts_in_message":[],"_revision_number":7},{"id":"bff6468f85ba158b294911640a0feee13b0a6dfe","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-20 10:32:53.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/7/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/11fe94a : FAILURE in 10m 06s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/7/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/4c79545 : FAILURE in 9m 55s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"31791b55f101a80f8f8d0bd68dac75fc5668095d","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-12-20 10:36:45.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"a632e55aec61659e1f71b7f9c186c183eb6bac53","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-20 10:46:34.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/851 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/682 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/452 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"120649428412cd9a66228c5a2a59378566ef6d04","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-20 12:25:31.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/7/check/tempest-dsvm-full-xenial/1377ea2/ : SUCCESS in 1h 07m 24s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/7/check/devstack-plugin-ceph-dsvm-xenial/ce4715a/ : SUCCESS in 1h 08m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a6aeee76a89aa901ed24061f337132dc94d63c36","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-20 12:51:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/7/check/dsvm-tempest-kvm/a9164f2 : SUCCESS in 1h 15m 31s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/7/check/dsvm-tempest-exe/e41503d : FAILURE in 47m 52s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"f4695ab6e08610ee82071949c650851bfbcb8e9e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-20 12:53:59.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/7/check/fujitsu-eternus-dx-iscsi/59e5e01 : ABORTED in 2h 30m 05s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/7/check/fujitsu-eternus-dx-fc/62dc851 : SUCCESS in 2h 28m 29s","accounts_in_message":[],"_revision_number":7},{"id":"32de7e32dc6ed41d0adf718b0ab8bb809d569fdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 12:59:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/7/check/openstack-tox-lower-constraints/1ec883a/ : FAILURE in 10m 20s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/7/check/openstack-tox-pep8/83126d5/ : SUCCESS in 9m 36s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/7/check/openstack-tox-py27/fadd7d9/ : FAILURE in 9m 18s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/7/check/openstack-tox-py35/5c1eccb/ : FAILURE in 8m 22s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/7/check/openstack-tox-py36/22b7030/ : FAILURE in 8m 00s\n- openstack-tox-docs http://logs.openstack.org/85/570885/7/check/openstack-tox-docs/02970fb/html/ : SUCCESS in 7m 57s\n- tempest-full http://logs.openstack.org/85/570885/7/check/tempest-full/0e6a806/ : SUCCESS in 1h 34m 12s\n- neutron-grenade http://logs.openstack.org/85/570885/7/check/neutron-grenade/c27339f/ : SUCCESS in 49m 40s\n- grenade-py3 http://logs.openstack.org/85/570885/7/check/grenade-py3/3275265/ : SUCCESS in 55m 26s\n- tempest-full-py3 http://logs.openstack.org/85/570885/7/check/tempest-full-py3/916c3df/ : SUCCESS in 1h 09m 12s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/7/check/cinder-tempest-dsvm-lvm-lio-barbican/3e85f22/ : SUCCESS in 1h 49m 24s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/7/check/cinder-tox-bandit-baseline/f9e7bea/ : SUCCESS in 6m 49s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/7/check/nova-multiattach/b5e8c61/ : SUCCESS in 54m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/7/check/openstack-tox-functional-py35/c614651/ : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/7/check/openstack-tox-functional-py36/3be045e/ : SUCCESS in 4m 47s\n- cinder-rally-task http://logs.openstack.org/85/570885/7/check/cinder-rally-task/00bdd2d/ : SUCCESS in 43m 27s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/7/check/openstack-tox-pylint/5852581/ : FAILURE in 5m 00s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/7/check/devstack-plugin-ceph-tempest/1657bd9/ : SUCCESS in 1h 46m 34s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/7/check/openstack-tox-functional/014bd83/ : SUCCESS in 4m 56s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/7/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/5e0cd25/ : SUCCESS in 1h 04m 28s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/7/check/legacy-tempest-dsvm-lvm-multibackend/484f814/ : SUCCESS in 1h 27m 33s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/7/check/legacy-tempest-dsvm-full-drbd-devstack/c7d493c/ : SUCCESS in 1h 08m 00s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/7/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/c7b612a/ : SUCCESS in 1h 18m 17s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/7/check/tempest-slow/57d7735/ : SUCCESS in 1h 52m 05s","accounts_in_message":[],"_revision_number":7},{"id":"e97c730221fd2b6a27faa1fad3de4f6b625d6f33","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-20 13:04:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/7/iscsi/ : SUCCESS in 58m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/7/smb/ : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":7},{"id":"75c6a996dd639161213bb209e37c83ceb3484d53","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-20 14:05:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/7/check/dsvm-tempest-cisco-zonemanager-job/211b937/ : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":7},{"id":"dcd1216e8d6ca0798ef24121ec3aac685b10b836","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-20 16:50:04.000000000","message":"Patch Set 6:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/85/570885/6/Lenovo-FC : SUCCESS in 35:11.36\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/85/570885/6/Lenovo-iSCSI : SUCCESS in 35:13.20","accounts_in_message":[],"_revision_number":6},{"id":"abc38df863db9f1818d1bdee93db77e591784420","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2018-12-21 00:04:36.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_FC http://publiclogs.emc.com/570885/7/EMC_VMAX_FC/693 : FAILURE in 36m 51s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/570885/7/EMC_VMAX_iSCSI/670 : FAILURE in 26m 26s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":7},{"id":"287e0300251105c6213d0b146c1b7b7dbac58dca","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-21 04:43:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/7/check/dsvm-tempest-storpool/d8e1aae : SUCCESS in 1h 40m 53s","accounts_in_message":[],"_revision_number":7},{"id":"e58612829508609dce7b80c8839b76517d9ae3a9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-21 06:03:35.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/85/570885/7/check/EMC_Unity_FC/ec26b5a/EMC_Unity_FC/596 : FAILURE in 1h 22m 24s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/7/check/EMC_Unity_iSCSI/dbf5177/EMC_Unity_iSCSI/564 : FAILURE in 1h 03m 35s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"614c9502ccaad5b5a77d1b19edf52f5ab4d7ed76","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2018-12-21 12:54:43.000000000","message":"Patch Set 7: Workflow-1\n\nBesides the unit test bug I\u0027m still in manual testing.","accounts_in_message":[],"_revision_number":7},{"id":"0651d9d0995d32554cb435a7fef84ba4b5e4e47c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-21 19:31:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/85/570885/7/check/dsvm-tempest-zadara-driver/c56d45d : FAILURE in 1h 52m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"564a5291f81e3656f48ecc21433fdea6b4f87d38","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-12-22 13:39:12.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12384 : SUCCESS in 50m 23s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/8802 : FAILURE in 58m 48s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10577 : FAILURE in 37m 50s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9183 : FAILURE in 42m 10s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10148 : FAILURE in 37m 48s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1198 : FAILURE in 1h 55m 42s","accounts_in_message":[],"_revision_number":7},{"id":"dc20d3c10ef8d6a5a362a16bf8a4be0a2e98baba","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-10 09:22:06.000000000","message":"Patch Set 7:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":7},{"id":"31d27382274579dd7eb3f8e39679d0cdca3c87c7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-10 10:22:23.000000000","message":"Patch Set 7:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"59a54538bc94623ff359c9d400c711a6e7327815","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-10 13:30:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1334601655a448db88f8ab75cf334992cd550c54","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-10 13:31:39.000000000","message":"Patch Set 8:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/8/check/dsvm-veritas-access-cinder-driver/b82c070 : FAILURE in 11s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/8/check/dsvm-veritas-access-cinder-iscsi-driver/21dddca : FAILURE in 11s","accounts_in_message":[],"_revision_number":8},{"id":"9fd7cb1c9a129fe25e3254f0eb4ac1f0ba815c6b","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-10 13:34:23.000000000","message":"Patch Set 8:\n\nNote on state: This is still a wacky attempt that works for the cloning but fails tempest tests during post-test-cleanup. Besides that the unit tests haven\u0027t been updated so far, so this is still open.\nReviews on the approach would already be interesting, thought. There\u0027s likely still a lot of work to do in order to make this nice \u0026 shiny.","accounts_in_message":[],"_revision_number":8},{"id":"2f086a2ba6b75c8d838ceb4afb75d6a1ba04e2f1","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-01-10 13:47:33.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/540 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/737 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/906 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"e2aca94d77d15dd5797d92ee448dafef070807c8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-10 14:23:12.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/8/iscsi/ : SUCCESS in 51m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/8/smb/ : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":8},{"id":"d64d3c5c8db08305c8fefb89275fb7ba7bafbbc2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-10 14:35:32.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/8/check/tempest-dsvm-full-xenial/43f0c81/ : FAILURE in 1h 03m 59s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/8/check/devstack-plugin-ceph-dsvm-xenial/993ea5a/ : FAILURE in 1h 04m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"685d9cbca9ed04b59981e8506fac56573a8c35d1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-10 14:41:50.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/8/check/dsvm-tempest-infinibox-fc/7eed315 : FAILURE in 55m 42s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/8/check/dsvm-tempest-infinibox-iscsi/c78b28e : FAILURE in 1h 10m 24s","accounts_in_message":[],"_revision_number":8},{"id":"89310f8f6af2ae2057d48da9470b34e091a2429a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-10 14:43:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/8/check/dsvm-tempest-kvm/9f072de : FAILURE in 1h 09m 27s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/8/check/dsvm-tempest-exe/c0167d4 : FAILURE in 55m 28s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"8ee1ac368d53a3667c4bbde9d851a252c91ec134","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-10 14:58:33.000000000","message":"Patch Set 8:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/8/check-cinder/Cinder-ISER-ISCSI/889113f : FAILURE in 53m 01s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/8/check-cinder/Cinder-ISER-LIO/128ab12 : FAILURE in 56m 33s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"3c7a4f9770d017624efddfa39390373bc5d261f7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-10 15:01:13.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/8/check/fujitsu-eternus-dx-iscsi/d702721 : FAILURE in 1h 29m 45s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/8/check/fujitsu-eternus-dx-fc/eba43da : FAILURE in 1h 26m 47s","accounts_in_message":[],"_revision_number":8},{"id":"55da382e015b432f219e4cb8d5200017a1d0f9e6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-10 15:02:50.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"b7b125a30a246884fc67be336fdd59bb8c6cd7b9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-10 15:10:31.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/8/check/dsvm-tempest-cisco-zonemanager-job/45c40e2/ : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":8},{"id":"067c88b0c0ba48043f9c844a48758ca7a24368f4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-10 15:20:14.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/8/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/4cb0cb0 : FAILURE in 1h 49m 07s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/8/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/00df660 : FAILURE in 1h 41m 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"57344042872c961b1f487cf6e4ed429d5bcef64c","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-01-10 15:21:06.000000000","message":"Patch Set 8:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://144.49.208.28:8000/build_logs/2019-01-10_06-30-42/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"e4d93a6b989b84becc976b16c00ce63c66ca0694","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-10 15:42:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/8/check/openstack-tox-lower-constraints/f6553d6/ : FAILURE in 10m 14s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/8/check/openstack-tox-pep8/5c86956/ : FAILURE in 7m 09s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/8/check/openstack-tox-py27/698d61f/ : FAILURE in 10m 17s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/8/check/openstack-tox-py35/e038e97/ : FAILURE in 7m 49s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/8/check/openstack-tox-py36/d6fd55b/ : FAILURE in 7m 24s\n- openstack-tox-docs http://logs.openstack.org/85/570885/8/check/openstack-tox-docs/273850b/html/ : SUCCESS in 7m 56s\n- tempest-full http://logs.openstack.org/85/570885/8/check/tempest-full/0bea916/ : FAILURE in 1h 01m 03s\n- neutron-grenade http://logs.openstack.org/85/570885/8/check/neutron-grenade/f7bedf4/ : FAILURE in 1h 00m 22s\n- grenade-py3 http://logs.openstack.org/85/570885/8/check/grenade-py3/2953228/ : FAILURE in 1h 38m 18s\n- tempest-full-py3 http://logs.openstack.org/85/570885/8/check/tempest-full-py3/72880a2/ : FAILURE in 56m 36s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/8/check/cinder-tempest-dsvm-lvm-lio-barbican/6d9eccf/ : FAILURE in 2h 08m 34s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/8/check/cinder-tox-bandit-baseline/75fd6a9/ : SUCCESS in 7m 00s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/8/check/nova-multiattach/11d36d4/ : FAILURE in 1h 14m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/8/check/openstack-tox-functional-py35/7ff5b6a/ : FAILURE in 5m 03s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/8/check/openstack-tox-functional-py36/de817b1/ : SUCCESS in 4m 54s\n- cinder-rally-task http://logs.openstack.org/85/570885/8/check/cinder-rally-task/899220f/ : FAILURE in 39m 15s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/8/check/openstack-tox-pylint/a677d68/ : FAILURE in 4m 37s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/8/check/devstack-plugin-ceph-tempest/f136705/ : TIMED_OUT in 2h 03m 19s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/8/check/openstack-tox-functional/8972aac/ : SUCCESS in 5m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/8/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/b95baa6/ : SUCCESS in 1h 20m 08s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/8/check/legacy-tempest-dsvm-lvm-multibackend/120066d/ : FAILURE in 1h 07m 52s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/8/check/legacy-tempest-dsvm-full-drbd-devstack/a48cec4/ : FAILURE in 1h 22m 02s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/8/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/40a7173/ : FAILURE in 1h 23m 34s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/8/check/tempest-slow/3feefce/ : FAILURE in 1h 48m 16s","accounts_in_message":[],"_revision_number":8},{"id":"1656d04ad21b505f2311d3578b75eb919de2dba6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-10 16:32:26.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/8/check-not-vote/ext-cinder-zuul/97b6dbb : FAILURE in 37m 42s","accounts_in_message":[],"_revision_number":8},{"id":"4ccb32b764c365349890d7272a842acd6a9b0fc2","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-10 17:43:51.000000000","message":"Patch Set 8:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/8/check/Cinder-AS13000/e074d0a : FAILURE in 2h 35m 07s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/8/check/cinder-test-fc/dd3aed8 : FAILURE in 56m 27s\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/8/check/cinder-test-iscsi/c2e1bd9 : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":8},{"id":"e13ed02ada660004bdff636027cb62ac17ba083a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-10 21:28:40.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/8/check/NEC-ISCSI/355e05e : FAILURE in 1h 31m 42s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/8/check/NEC-FC/46791f8 : FAILURE in 1h 28m 41s","accounts_in_message":[],"_revision_number":8},{"id":"ad164bd2500f691a30d7b83fc15f2c905c81f9ff","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-11 07:08:16.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/655 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"b9d42b73a075dc9eec267c54f70b7ea43b30b25a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-01-11 20:07:34.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_FC http://publiclogs.emc.com/570885/8/EMC_VMAX_FC/764 : FAILURE in 7h 02m 07s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/570885/8/EMC_VMAX_iSCSI/751 : FAILURE in 6h 59m 29s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":8},{"id":"d95c7329457428ef368565bee58ac384486e40a2","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-23 08:37:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b7e6e9dc0744a07f57a08266bbe33b3aba7e4bc9","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-23 08:38:05.000000000","message":"Patch Set 9:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/9/check/dsvm-veritas-access-cinder-driver/b4bf475 : FAILURE in 12s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/9/check/dsvm-veritas-access-cinder-iscsi-driver/6450b5b : FAILURE in 12s","accounts_in_message":[],"_revision_number":9},{"id":"6125754f73b466a74147c4285968c57f59ba8222","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-01-23 09:07:42.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/85/570885/9/check/EMC_Unity_FC/8e2af17/EMC_Unity_FC/871 : FAILURE in 24m 28s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/9/check/EMC_Unity_iSCSI/512a803/EMC_Unity_iSCSI/837 : FAILURE in 22m 45s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"322f7dc616e453dc72c552d413de714b13e5ea03","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-23 09:40:31.000000000","message":"Patch Set 9:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/9/check-cinder/Cinder-ISER-ISCSI/73e7609 : FAILURE in 55m 14s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/9/check-cinder/Cinder-ISER-LIO/b48fbe2 : FAILURE in 58m 32s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"275152dc3f6fe9ee05bb8555adf83d3e0293d4dc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-23 09:54:25.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/9/check/tempest-dsvm-full-xenial/4d42712/ : FAILURE in 58m 46s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/9/check/devstack-plugin-ceph-dsvm-xenial/d33d575/ : FAILURE in 1h 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d19681b5fd0071939f81ae108e9b3ceace44a592","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-23 09:57:25.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/9/check/dsvm-tempest-kvm/dcbc365 : FAILURE in 1h 09m 45s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/9/check/dsvm-tempest-exe/a6388aa : SUCCESS in 52m 30s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"7f00cb7116af972aff1935c7b255ce10b27d04dd","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-23 09:59:01.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/9/iscsi/ : FAILURE in 1h 21m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/9/smb/ : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":9},{"id":"dffa6a58ba3e2d6988d16cb0f8c19ab9fce0e9c8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-23 10:17:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/9/check/dsvm-tempest-cisco-zonemanager-job/73c12c8/ : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":9},{"id":"c0199e5acc2c729ea163f048c714cb3be93853b5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-23 10:21:50.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/9/check/dsvm-tempest-infinibox-fc/cfbffa8 : FAILURE in 52m 11s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/9/check/dsvm-tempest-infinibox-iscsi/35070cc : FAILURE in 1h 12m 13s","accounts_in_message":[],"_revision_number":9},{"id":"3f8125c189ec686b7f9d2caa9ad9df7e40427977","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-23 10:45:44.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"0744960f28f01cf4e4e5cf399c52a1459ffbdd90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-23 10:54:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/9/check/openstack-tox-lower-constraints/8a8040e/ : FAILURE in 11m 12s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/9/check/openstack-tox-pep8/25bb44b/ : FAILURE in 9m 13s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/9/check/openstack-tox-py27/ee5ff92/ : FAILURE in 11m 30s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/9/check/openstack-tox-py35/48e8f66/ : FAILURE in 11m 38s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/9/check/openstack-tox-py36/0f14349/ : FAILURE in 9m 46s\n- openstack-tox-docs http://logs.openstack.org/85/570885/9/check/openstack-tox-docs/1ba57c3/html/ : SUCCESS in 10m 24s\n- tempest-full http://logs.openstack.org/85/570885/9/check/tempest-full/761ee42/ : RETRY_LIMIT in 3m 04s\n- neutron-grenade http://logs.openstack.org/85/570885/9/check/neutron-grenade/2c46b1a/ : FAILURE in 59m 03s\n- grenade-py3 http://logs.openstack.org/85/570885/9/check/grenade-py3/24fffda/ : FAILURE in 1h 11m 29s\n- tempest-full-py3 http://logs.openstack.org/85/570885/9/check/tempest-full-py3/8d3d735/ : FAILURE in 19m 06s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/9/check/cinder-tempest-dsvm-lvm-lio-barbican/1abdd97/ : FAILURE in 21m 54s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/9/check/cinder-tox-bandit-baseline/c16adcd/ : SUCCESS in 7m 31s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/9/check/nova-multiattach/7e5f67e/ : FAILURE in 1h 19m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/9/check/openstack-tox-functional-py35/cf6cae1/ : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/9/check/openstack-tox-functional-py36/57872b6/ : FAILURE in 5m 12s\n- cinder-rally-task http://logs.openstack.org/85/570885/9/check/cinder-rally-task/9aed33c/ : FAILURE in 1h 04m 43s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/9/check/openstack-tox-pylint/4d880a6/ : FAILURE in 4m 52s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/9/check/devstack-plugin-ceph-tempest/4f7d2f3/ : TIMED_OUT in 2h 07m 06s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/9/check/openstack-tox-functional/ffbf831/ : SUCCESS in 5m 25s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/9/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/c24b99e/ : FAILURE in 59m 01s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/9/check/legacy-tempest-dsvm-lvm-multibackend/03812a8/ : FAILURE in 1h 31m 56s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/9/check/legacy-tempest-dsvm-full-drbd-devstack/8676c88/ : FAILURE in 1h 30m 07s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/9/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/cb2086a/ : FAILURE in 20m 38s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/9/check/tempest-slow/67af42b/ : FAILURE in 2h 08m 47s","accounts_in_message":[],"_revision_number":9},{"id":"2b4d68611feadfa6747ec365da803fd87838753d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-23 11:12:30.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/9/check/fujitsu-eternus-dx-iscsi/bf00f57 : ABORTED in 2h 30m 03s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/9/check/fujitsu-eternus-dx-fc/968cf7b : FAILURE in 36m 59s","accounts_in_message":[],"_revision_number":9},{"id":"5a898e5e464661867dfbd2fdc6dfcbef40f358c5","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-23 11:32:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bbbc11cef5c617ca8cdc1248aa30a49e157eaa06","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-23 11:32:54.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/10/check/dsvm-veritas-access-cinder-driver/a4b93ec : FAILURE in 12s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/10/check/dsvm-veritas-access-cinder-iscsi-driver/55ea80e : FAILURE in 12s","accounts_in_message":[],"_revision_number":10},{"id":"467cf545c11b8c2fd8bd40122c85168aac2a0777","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-01-23 11:58:14.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/85/570885/10/check/EMC_Unity_FC/84b5f90/EMC_Unity_FC/873 : FAILURE in 25m 56s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/10/check/EMC_Unity_iSCSI/25e3826/EMC_Unity_iSCSI/839 : FAILURE in 22m 38s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"2e9ff68dc1d00c517b1219fdc0f8a626bf464d07","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-23 12:25:56.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/10/iscsi/ : SUCCESS in 53m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/10/smb/ : FAILURE in 48m 34s","accounts_in_message":[],"_revision_number":10},{"id":"ca41f3c803e1b29874ed8de8f4d946bc9f22e1da","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-23 12:32:01.000000000","message":"Patch Set 10:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/10/check-cinder/Cinder-ISER-ISCSI/2f24073 : FAILURE in 58m 57s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/10/check-cinder/Cinder-ISER-LIO/05b38bd : FAILURE in 56m 46s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"fa00470ccde77d15826b008405408ca515dca9f8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-23 12:43:11.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/10/check/tempest-dsvm-full-xenial/e816f37/ : FAILURE in 1h 10m 47s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/10/check/devstack-plugin-ceph-dsvm-xenial/bd11876/ : FAILURE in 1h 06m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"05da669dc0ed62961bb8c143364edaea5632af32","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-23 12:46:06.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/10/check/dsvm-tempest-cisco-zonemanager-job/4a9b975/ : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":10},{"id":"470ffef5bea1041f3d71cfbd6f4d7245fa814e87","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-23 12:53:40.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/10/check/dsvm-tempest-infinibox-fc/2064aeb : FAILURE in 1h 20m 11s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/10/check/dsvm-tempest-infinibox-iscsi/422ba74 : FAILURE in 1h 20m 50s","accounts_in_message":[],"_revision_number":10},{"id":"98367fa72fa545811bb54f7c9253eef81321c324","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-23 13:02:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/10/check/fujitsu-eternus-dx-iscsi/137a2e6 : FAILURE in 1h 29m 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/10/check/fujitsu-eternus-dx-fc/661a329 : FAILURE in 1h 25m 35s","accounts_in_message":[],"_revision_number":10},{"id":"91baacb7ac073bc17b6b1a6f963b7af31dfa4e49","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-01-23 13:03:00.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/85/570885/10/check/EMC_XtremIO_FC/0fdd345/EMC_XtremIO_FC/1301 : FAILURE in 1h 30m 36s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/10/check/EMC_XtremIO_iSCSI/e3b681a/EMC_XtremIO_iSCSI/1326 : FAILURE in 1h 17m 27s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":10},{"id":"aa8f118cc37f527bea4766f403b27a72f4819ab4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-23 13:04:11.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"c7917f5e51d9bffadd8268699463646f921d5049","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-23 13:09:22.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/10/check/dsvm-tempest-kvm/adb68b0 : FAILURE in 1h 05m 50s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/10/check/dsvm-tempest-exe/72a899a : FAILURE in 1h 13m 30s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"f4fbb99598c907e4ae38d2ed942d3e20c5a3cd30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-23 13:47:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/10/check/openstack-tox-lower-constraints/941cd1a/ : FAILURE in 12m 15s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/10/check/openstack-tox-pep8/5dee4de/ : SUCCESS in 7m 22s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/10/check/openstack-tox-py27/ae0df41/ : FAILURE in 9m 14s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/10/check/openstack-tox-py35/40f9fbc/ : FAILURE in 9m 28s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/10/check/openstack-tox-py36/d9f1acc/ : FAILURE in 8m 22s\n- openstack-tox-docs http://logs.openstack.org/85/570885/10/check/openstack-tox-docs/01e0899/html/ : SUCCESS in 6m 47s\n- tempest-full http://logs.openstack.org/85/570885/10/check/tempest-full/520003f/ : FAILURE in 1h 21m 38s\n- neutron-grenade http://logs.openstack.org/85/570885/10/check/neutron-grenade/c109980/ : FAILURE in 1h 01m 02s\n- grenade-py3 http://logs.openstack.org/85/570885/10/check/grenade-py3/7c79618/ : FAILURE in 1h 08m 44s\n- tempest-full-py3 http://logs.openstack.org/85/570885/10/check/tempest-full-py3/c522796/ : FAILURE in 1h 16m 17s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/10/check/cinder-tempest-dsvm-lvm-lio-barbican/d4a0ff4/ : FAILURE in 1h 42m 50s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/10/check/cinder-tox-bandit-baseline/ae51d24/ : SUCCESS in 6m 43s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/10/check/nova-multiattach/e34d9e1/ : FAILURE in 18m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/10/check/openstack-tox-functional-py35/46d5806/ : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/10/check/openstack-tox-functional-py36/c247414/ : FAILURE in 4m 33s\n- cinder-rally-task http://logs.openstack.org/85/570885/10/check/cinder-rally-task/46f7108/ : FAILURE in 37m 59s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/10/check/openstack-tox-pylint/33b44e4/ : FAILURE in 4m 35s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/10/check/devstack-plugin-ceph-tempest/e3325d5/ : FAILURE in 1h 53m 50s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/10/check/openstack-tox-functional/43a0d28/ : FAILURE in 5m 42s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/10/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/a52e851/ : SUCCESS in 1h 06m 15s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/10/check/legacy-tempest-dsvm-lvm-multibackend/bc1c518/ : FAILURE in 1h 29m 35s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/10/check/legacy-tempest-dsvm-full-drbd-devstack/2cc7bcb/ : FAILURE in 19m 52s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/10/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/e248ea5/ : FAILURE in 1h 23m 35s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/10/check/tempest-slow/6866fe1/ : FAILURE in 2h 12m 37s","accounts_in_message":[],"_revision_number":10},{"id":"dee488a672c4e7290dfcb73ac1c6d9f6216eae39","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-23 20:11:15.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/10/check-not-vote/ext-cinder-zuul/a6cf4d4 : FAILURE in 17m 18s","accounts_in_message":[],"_revision_number":10},{"id":"85a4f6f94ef62f6d37059031c7d2a9a4dcaed899","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-01-24 00:41:52.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-storpoolci\u0027.  For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/StorPool_CI\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/10/check/dsvm-tempest-storpool/2a24f73 : FAILURE in 1h 37m 14s","accounts_in_message":[],"_revision_number":10},{"id":"be2d88fff357515e33d4f3a2223ef0bed391381b","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-24 01:58:08.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/10/check/Cinder-AS13000/c5988dd : FAILURE in 2h 18m 17s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/10/check/cinder-test-fc/9b5c944 : FAILURE in 59m 01s\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/10/check/cinder-test-iscsi/81e330c : FAILURE in 1h 00m 55s","accounts_in_message":[],"_revision_number":10},{"id":"2dd03c1f05c29f68a10f76aae1e1d2a3563dfe27","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-24 02:00:55.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/10/check/NEC-ISCSI/1661716 : FAILURE in 1h 45m 27s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/10/check/NEC-FC/311a5b9 : FAILURE in 1h 48m 42s","accounts_in_message":[],"_revision_number":10},{"id":"df76daabacef280cee13c03fe22273c5b5fcc3fa","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-01-24 06:09:41.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)\n\nAll the create volume tests (clone, from image, from source_vol ...) needs to be modified (adding the context parameter).","accounts_in_message":[],"_revision_number":10},{"id":"798a5d62c8238f276be0985ff59beb881a7a4b2b","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-24 08:02:21.000000000","message":"Patch Set 10: Workflow-1\n\nThanks for the feedback.\nUnit tests are not updated so far as i\u0027m still looking into a tempest test volume cleanup issue. Once that is done i\u0027ll update the unit tests go out of WIP state.","accounts_in_message":[],"_revision_number":10},{"id":"40676d768a578b60dc20cbe3bb9bbb4d97f48122","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-24 10:42:41.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"895d97a9f3aaf982ed79ff4695d249df725131c2","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-24 10:43:27.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"8376ab7d94c1013766484d6b2992aa1860e10d82","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-24 10:47:57.000000000","message":"Patch Set 11:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/11/check/dsvm-veritas-access-cinder-driver/0bd0c65 : FAILURE in 31s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/11/check/dsvm-veritas-access-cinder-iscsi-driver/6635172 : FAILURE in 12s","accounts_in_message":[],"_revision_number":11},{"id":"37a75671160495603bbd569a6d4e3f9de6a90b7c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-24 11:00:37.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"ffed66d1b4649eea770dbf5cc9f5592f4034e544","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-24 11:34:31.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/11/iscsi/ : SUCCESS in 50m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/11/smb/ : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":11},{"id":"36d6fbf257df0d6c767f9742b490aae7b784388e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-24 11:54:33.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/11/check/dsvm-tempest-infinibox-fc/4b4bcd6 : FAILURE in 1h 03m 13s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/11/check/dsvm-tempest-infinibox-iscsi/2adfe05 : FAILURE in 1h 11m 04s","accounts_in_message":[],"_revision_number":11},{"id":"392f370eb1c7651491e630620aae16b081db8cf3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-24 12:01:50.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/11/check/tempest-dsvm-full-xenial/96ae622/ : FAILURE in 1h 03m 23s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/11/check/devstack-plugin-ceph-dsvm-xenial/57b6efa/ : FAILURE in 1h 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"90e3725ce8578e33b5e61d807f2fd5c87ac86d98","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-24 12:13:56.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/11/check/dsvm-tempest-cisco-zonemanager-job/a5bf4d3/ : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":11},{"id":"841cb439f34a94fb6c53b2ffffef97ae73f35080","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-24 12:37:05.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/11/check/fujitsu-eternus-dx-iscsi/ddfff88 : FAILURE in 1h 53m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/11/check/fujitsu-eternus-dx-fc/d1c4bc4 : FAILURE in 1h 53m 28s","accounts_in_message":[],"_revision_number":11},{"id":"74febde237fa090f8f9d04e5db63b273a45a5a85","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-24 12:37:49.000000000","message":"Patch Set 11:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/11/check-cinder/Cinder-ISER-ISCSI/4994bc4 : FAILURE in 56m 40s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/11/check-cinder/Cinder-ISER-LIO/d78ca7a : FAILURE in 57m 00s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"43e47971819089fdd0894acbcb98635ee36aa08a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-24 12:37:50.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/11/check/dsvm-tempest-kvm/9a0edc2 : FAILURE in 1h 27m 25s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/11/check/dsvm-tempest-exe/139949a : FAILURE in 50m 35s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"65e4b9786ea00b28ac0263f8895d95173f7ccc33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-24 13:00:40.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/11/check/openstack-tox-lower-constraints/10d9912/ : FAILURE in 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/11/check/openstack-tox-pep8/44d6666/ : SUCCESS in 9m 01s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/11/check/openstack-tox-py27/b0a8fbb/ : FAILURE in 11m 11s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/11/check/openstack-tox-py35/217243b/ : FAILURE in 11m 10s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/11/check/openstack-tox-py36/f5ae2e8/ : FAILURE in 10m 27s\n- openstack-tox-docs http://logs.openstack.org/85/570885/11/check/openstack-tox-docs/02bf98b/html/ : SUCCESS in 8m 29s\n- tempest-full http://logs.openstack.org/85/570885/11/check/tempest-full/0a53c51/ : FAILURE in 1h 28m 34s\n- neutron-grenade http://logs.openstack.org/85/570885/11/check/neutron-grenade/2d890b7/ : FAILURE in 53m 13s\n- grenade-py3 http://logs.openstack.org/85/570885/11/check/grenade-py3/ae3f07c/ : FAILURE in 1h 14m 06s\n- tempest-full-py3 http://logs.openstack.org/85/570885/11/check/tempest-full-py3/11d8ff0/ : FAILURE in 1h 27m 28s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/11/check/cinder-tempest-dsvm-lvm-lio-barbican/9e8fe11/ : FAILURE in 1h 57m 36s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/11/check/cinder-tox-bandit-baseline/ac50005/ : SUCCESS in 9m 42s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/11/check/nova-multiattach/f249fab/ : FAILURE in 1h 17m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/11/check/openstack-tox-functional-py35/56c34ad/ : FAILURE in 5m 07s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/11/check/openstack-tox-functional-py36/4faa10a/ : FAILURE in 4m 39s\n- cinder-rally-task http://logs.openstack.org/85/570885/11/check/cinder-rally-task/1703332/ : FAILURE in 42m 45s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/11/check/openstack-tox-pylint/75ff24a/ : FAILURE in 4m 58s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/11/check/devstack-plugin-ceph-tempest/6fe9308/ : FAILURE in 2h 02m 41s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/11/check/openstack-tox-functional/b255a8d/ : FAILURE in 7m 35s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/11/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/a4daaba/ : SUCCESS in 1h 05m 59s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/11/check/legacy-tempest-dsvm-lvm-multibackend/063a3f2/ : FAILURE in 2h 10m 36s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/11/check/legacy-tempest-dsvm-full-drbd-devstack/9f29846/ : FAILURE in 1h 22m 37s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/11/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/9648cfb/ : FAILURE in 1h 24m 35s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/11/check/tempest-slow/ca24e8e/ : FAILURE in 2h 01m 09s","accounts_in_message":[],"_revision_number":11},{"id":"2c714b3898e77c49f6bfbcd2656f6d010448e8e4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-24 13:14:43.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/11/check-not-vote/ext-cinder-zuul/402cf8f : FAILURE in 39m 06s","accounts_in_message":[],"_revision_number":11},{"id":"4add6b2a8bc28fda6943e828c3b9a1d8d25dc4f9","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-24 19:03:20.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/913 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"1a4538875a85113a3e5fadf5c53f653dca1f5d75","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-01-24 21:22:09.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-storpoolci\u0027.  For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/StorPool_CI\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/11/check/dsvm-tempest-storpool/46b8a8a : FAILURE in 1h 38m 29s","accounts_in_message":[],"_revision_number":11},{"id":"441b9a044c3777787c0b35db8759920f6cdb3d78","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-01-24 23:12:13.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/85/570885/11/check/EMC_XtremIO_FC/5e97a4b/EMC_XtremIO_FC/1323 : FAILURE in 1h 39m 53s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/11/check/EMC_XtremIO_iSCSI/a2f2893/EMC_XtremIO_iSCSI/1348 : FAILURE in 1h 21m 19s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":11},{"id":"5a9df9352a7a6a4ef9d39cddaeb78d7ec6d1307c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-25 00:34:31.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/11/check/NEC-ISCSI/1f1fad0 : FAILURE in 1h 13m 53s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/11/check/NEC-FC/dfb1fb6 : FAILURE in 1h 49m 55s","accounts_in_message":[],"_revision_number":11},{"id":"11a14e02b0964fd318e6ac5bf54236fad3f4bddf","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-25 08:44:37.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"4b5a120f905c81e604fb9d17808816bf8fdae6e9","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-25 08:50:53.000000000","message":"Patch Set 12:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/12/check/dsvm-veritas-access-cinder-driver/1aa322e : FAILURE in 31s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/12/check/dsvm-veritas-access-cinder-iscsi-driver/97607f2 : FAILURE in 30s","accounts_in_message":[],"_revision_number":12},{"id":"9c768d0001e9cc6ed91e51acbb070a6c648abf81","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-25 09:16:05.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"3886a31eda05eb3fb42f020396fbbb0219d23b7c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-25 09:20:06.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/12/check/dsvm-tempest-infinibox-fc/10e20d9 : FAILURE in 9m 04s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/12/check/dsvm-tempest-infinibox-iscsi/1682543 : FAILURE in 17m 52s","accounts_in_message":[],"_revision_number":12},{"id":"4b56a555513b6687d0cc49ab10ff127efb0acfe7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-25 09:43:44.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/12/iscsi/ : SUCCESS in 58m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/12/smb/ : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":12},{"id":"6f4a0d77764c21658608461948f998f10e4693b8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-25 09:55:59.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/12/check/tempest-dsvm-full-xenial/37dc10b/ : FAILURE in 1h 04m 12s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/12/check/devstack-plugin-ceph-dsvm-xenial/2b9d662/ : FAILURE in 1h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c6bc0825250fd5d4fcb0f960228c19896262af2f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-25 10:02:05.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/12/check-not-vote/ext-cinder-zuul/e5f5c2b : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":12},{"id":"4a2630eda0f85f3a7e9618fe79ef41897c90f49a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-25 10:46:00.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/12/check/dsvm-tempest-kvm/cb6bcc3 : FAILURE in 1h 11m 51s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/12/check/dsvm-tempest-exe/7663f79 : SUCCESS in 54m 40s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"f3b670e88927623a87fddf3663f6a1c7ef34bba3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 11:07:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/12/check/openstack-tox-lower-constraints/bc4de78/ : FAILURE in 10m 12s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/12/check/openstack-tox-pep8/43ca9f2/ : SUCCESS in 8m 15s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/12/check/openstack-tox-py27/b372670/ : FAILURE in 10m 55s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/12/check/openstack-tox-py35/b4e6120/ : FAILURE in 9m 42s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/12/check/openstack-tox-py36/79b6301/ : FAILURE in 8m 06s\n- openstack-tox-docs http://logs.openstack.org/85/570885/12/check/openstack-tox-docs/bc3beba/html/ : SUCCESS in 8m 58s\n- tempest-full http://logs.openstack.org/85/570885/12/check/tempest-full/00b78e0/ : FAILURE in 1h 27m 36s\n- neutron-grenade http://logs.openstack.org/85/570885/12/check/neutron-grenade/2e1f4ae/ : FAILURE in 1h 04m 27s\n- grenade-py3 http://logs.openstack.org/85/570885/12/check/grenade-py3/192d95a/ : FAILURE in 1h 01m 05s\n- tempest-full-py3 http://logs.openstack.org/85/570885/12/check/tempest-full-py3/e040723/ : FAILURE in 1h 06m 36s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/12/check/cinder-tempest-dsvm-lvm-lio-barbican/0e3b445/ : FAILURE in 1h 49m 30s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/12/check/cinder-tox-bandit-baseline/27365e9/ : SUCCESS in 7m 50s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/12/check/nova-multiattach/c7b1737/ : FAILURE in 1h 09m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/12/check/openstack-tox-functional-py35/090668f/ : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/12/check/openstack-tox-functional-py36/5b35ed0/ : SUCCESS in 5m 58s\n- cinder-rally-task http://logs.openstack.org/85/570885/12/check/cinder-rally-task/2805325/ : FAILURE in 39m 40s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/12/check/openstack-tox-pylint/64a0f20/ : FAILURE in 5m 24s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/12/check/devstack-plugin-ceph-tempest/d002479/ : FAILURE in 2h 01m 13s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/12/check/openstack-tox-functional/10a4c18/ : SUCCESS in 5m 38s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/12/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/3537d2d/ : SUCCESS in 1h 01m 21s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/12/check/legacy-tempest-dsvm-lvm-multibackend/584b283/ : FAILURE in 1h 28m 33s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/12/check/legacy-tempest-dsvm-full-drbd-devstack/2238c29/ : FAILURE in 1h 21m 48s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/12/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/062d82e/ : FAILURE in 1h 24m 40s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/12/check/tempest-slow/9895d3d/ : FAILURE in 2h 09m 56s","accounts_in_message":[],"_revision_number":12},{"id":"757a2a54c9f059035a5bb93cb24474aa9a0edb9f","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-25 11:25:45.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"5a424d024049447e6edbf0cd309993b4dea695eb","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-25 11:26:59.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/923 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"0c45b9390b3767c32b3d440ba30e1bbdf48b8e70","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-25 11:32:11.000000000","message":"Patch Set 13:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/13/check/dsvm-veritas-access-cinder-driver/edf4c72 : FAILURE in 13s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/13/check/dsvm-veritas-access-cinder-iscsi-driver/eda0507 : FAILURE in 31s","accounts_in_message":[],"_revision_number":13},{"id":"bab0f548f8787041517edc9f59884ea5097fa9e4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-25 11:43:35.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"a64b3a5247ca1fc81e9f13e510223c177dbe72d6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-25 11:45:18.000000000","message":"Patch Set 13:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/13/check/dsvm-tempest-infinibox-fc/1336518 : FAILURE in 11m 01s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/13/check/dsvm-tempest-infinibox-iscsi/b6e192a : FAILURE in 17m 29s","accounts_in_message":[],"_revision_number":13},{"id":"74c93460faf88dd32b8be4c436c23006d16bb160","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-25 11:59:12.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/13/check-not-vote/ext-cinder-zuul/1375985 : FAILURE in 32m 05s","accounts_in_message":[],"_revision_number":13},{"id":"8ed3b669bcfbb77570d74f2aff04bff3ae900870","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-25 12:30:26.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/13/iscsi/ : SUCCESS in 56m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/13/smb/ : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":13},{"id":"c9a6f3d25470367ba2647782aeff1a4854a2feff","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-25 12:37:34.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/13/check/tempest-dsvm-full-xenial/0f940d6/ : FAILURE in 1h 08m 14s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/13/check/devstack-plugin-ceph-dsvm-xenial/2d1e6d2/ : FAILURE in 1h 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a864fe4af903c8a0ad79accae2c06665583a53ee","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-25 12:41:26.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/13/check/dsvm-tempest-cisco-zonemanager-job/c6f14c7/ : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":13},{"id":"e194353ac783f3b22131f702a368f32e6336686d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-25 13:18:38.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/13/check/dsvm-tempest-kvm/2e124e6 : FAILURE in 1h 06m 22s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/13/check/dsvm-tempest-exe/de801dc : SUCCESS in 59m 57s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"1ad7c68f7d2714d0eb2e3bd45607686b515b24ed","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-25 13:33:33.000000000","message":"Patch Set 13:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/13/check/fujitsu-eternus-dx-iscsi/5f27f0c : FAILURE in 2h 04m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/13/check/fujitsu-eternus-dx-fc/37a1b35 : FAILURE in 2h 02m 37s","accounts_in_message":[],"_revision_number":13},{"id":"5db6f31bcfa9a7afb1e7d0eeec9718dab67eb9b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 14:01:58.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/13/check/openstack-tox-lower-constraints/a67aab5/ : FAILURE in 16m 31s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/13/check/openstack-tox-pep8/a20ad2f/ : SUCCESS in 7m 26s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/13/check/openstack-tox-py27/ca36f18/ : FAILURE in 8m 59s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/13/check/openstack-tox-py35/9405403/ : FAILURE in 8m 26s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/13/check/openstack-tox-py36/638ea89/ : FAILURE in 9m 29s\n- openstack-tox-docs http://logs.openstack.org/85/570885/13/check/openstack-tox-docs/04de978/html/ : SUCCESS in 8m 43s\n- tempest-full http://logs.openstack.org/85/570885/13/check/tempest-full/e5d0435/ : FAILURE in 1h 47m 24s\n- neutron-grenade http://logs.openstack.org/85/570885/13/check/neutron-grenade/a31535a/ : FAILURE in 47m 35s\n- grenade-py3 http://logs.openstack.org/85/570885/13/check/grenade-py3/058be26/ : FAILURE in 58m 09s\n- tempest-full-py3 http://logs.openstack.org/85/570885/13/check/tempest-full-py3/fa3c6bf/ : FAILURE in 59m 24s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/13/check/cinder-tempest-dsvm-lvm-lio-barbican/f845493/ : FAILURE in 2h 28m 27s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/13/check/cinder-tox-bandit-baseline/d1f51f9/ : SUCCESS in 7m 04s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/13/check/nova-multiattach/32d50f4/ : FAILURE in 1h 10m 55s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/13/check/openstack-tox-functional-py35/f255847/ : FAILURE in 5m 08s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/13/check/openstack-tox-functional-py36/9e665d9/ : FAILURE in 5m 58s\n- cinder-rally-task http://logs.openstack.org/85/570885/13/check/cinder-rally-task/c5a1678/ : FAILURE in 44m 17s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/13/check/openstack-tox-pylint/64dd581/ : FAILURE in 6m 39s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/13/check/devstack-plugin-ceph-tempest/eb60052/ : FAILURE in 2h 02m 28s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/13/check/openstack-tox-functional/9972ffe/ : SUCCESS in 7m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/13/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/e41075e/ : SUCCESS in 1h 08m 12s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/13/check/legacy-tempest-dsvm-lvm-multibackend/eeaa718/ : FAILURE in 1h 27m 45s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/13/check/legacy-tempest-dsvm-full-drbd-devstack/599b4a6/ : FAILURE in 1h 18m 14s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/13/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/b5ff393/ : FAILURE in 1h 12m 44s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/13/check/tempest-slow/472bb8d/ : FAILURE in 2h 04m 30s","accounts_in_message":[],"_revision_number":13},{"id":"6e1860b64e615eb471ba5bbc3be9d8570dc1c457","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-25 15:09:49.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/927 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"e8dd8d03b6dbb300622c7ac1fc65bd4fef5144c5","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-25 15:14:03.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"b9751f48e70f74ea96e99c730fc7529f83ef5395","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-01-25 17:06:11.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/85/570885/13/check/EMC_XtremIO_FC/e2d17e9/EMC_XtremIO_FC/1334 : FAILURE in 2h 26m 17s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/13/check/EMC_XtremIO_iSCSI/3ed0141/EMC_XtremIO_iSCSI/1363 : FAILURE in 1h 20m 39s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":13},{"id":"b8c51f355479b05277aabf1316a0d874a0bd57c5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-01-25 17:45:54.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/85/570885/13/check/EMC_Unity_FC/d945b36/EMC_Unity_FC/910 : FAILURE in 1h 20m 25s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/13/check/EMC_Unity_iSCSI/0d78308/EMC_Unity_iSCSI/872 : FAILURE in 1h 48m 38s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"75338653ab5f9a292079dbd7a08e1eb5fe73d6bc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-01-25 20:05:16.000000000","message":"Patch Set 13:\n\nBuild failed.  To recheck use \u0027run-storpoolci\u0027.  For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/StorPool_CI\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/13/check/dsvm-tempest-storpool/105380d : FAILURE in 1h 36m 56s","accounts_in_message":[],"_revision_number":13},{"id":"bf9e6f3bd548dbf08de6b6d1b6a3931d762c422f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-25 23:14:28.000000000","message":"Patch Set 13:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/13/check/NEC-ISCSI/4af270b : FAILURE in 1h 48m 12s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/13/check/NEC-FC/3c0fba1 : FAILURE in 1h 45m 44s","accounts_in_message":[],"_revision_number":13},{"id":"01e5c362e42af95360bebce2296f4458fbbcc0b6","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-01-27 04:23:06.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12672 : FAILURE in 36m 35s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9038 : FAILURE in 1h 08m 34s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10796 : FAILURE in 56m 29s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9416 : FAILURE in 1h 40m 13s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10382 : FAILURE in 9m 47s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1353 : FAILURE in 1h 57m 48s","accounts_in_message":[],"_revision_number":13},{"id":"6e854f47dd8301aeec6ec098afa7ce5d449df42c","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-30 14:38:03.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3b88d390d5d7009cec5b5af636035f2937d9c3bf","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-30 14:39:14.000000000","message":"Patch Set 14:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/14/check/dsvm-veritas-access-cinder-driver/faaa1ed : FAILURE in 32s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/14/check/dsvm-veritas-access-cinder-iscsi-driver/70944e0 : FAILURE in 32s","accounts_in_message":[],"_revision_number":14},{"id":"a280164ecc7a9831966d75edb7ed490ec6b98a32","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-30 14:55:38.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"4ba42598eabc0ce524bda2ead7f0c83610bb3f68","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-01-30 14:58:01.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/716 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1055 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/894 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"6795f3069dcabecac91feed9aa8fd111b70118be","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-30 15:40:17.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/14/iscsi/ : SUCCESS in 1h 01m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/14/smb/ : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":14},{"id":"124f2c1b46e28c15993e755a1c870aa1bdb41c46","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-30 15:46:02.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/14/check/tempest-dsvm-full-xenial/f77e0bd/ : FAILURE in 1h 03m 11s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/14/check/devstack-plugin-ceph-dsvm-xenial/a240550/ : FAILURE in 1h 07m 40s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7a494eada037b5fe8eaed05372e40d06bd47cfab","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-30 15:58:27.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/14/check-not-vote/ext-cinder-zuul/67ccb67 : FAILURE in 35m 27s","accounts_in_message":[],"_revision_number":14},{"id":"ef4b28427b7e79afed466830166a9840040b8b08","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-30 16:12:53.000000000","message":"Patch Set 14:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/14/check/fujitsu-eternus-dx-iscsi/071ebcd : FAILURE in 1h 30m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/14/check/fujitsu-eternus-dx-fc/02c59d2 : FAILURE in 1h 33m 12s","accounts_in_message":[],"_revision_number":14},{"id":"7abad246b897507e813c4cc154de33a668df10ee","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-30 16:21:27.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/14/check/dsvm-tempest-infinibox-fc/47a3d45 : FAILURE in 53m 14s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/14/check/dsvm-tempest-infinibox-iscsi/0613624 : FAILURE in 1h 22m 02s","accounts_in_message":[],"_revision_number":14},{"id":"f0068b81ad83848109db1f984e802a0fb33058e2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-30 17:04:46.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/14/check/dsvm-tempest-cisco-zonemanager-job/91987d2/ : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":14},{"id":"b1bbf811ee79d427d55028d505e907180ada8f91","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-30 17:29:40.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/14/check/dsvm-tempest-kvm/592370f : FAILURE in 1h 05m 13s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/14/check/dsvm-tempest-exe/31f62ea : FAILURE in 10m 53s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":14},{"id":"6456f3a40ebcfeeea1a0b9997d31f7736d4570f7","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-30 17:42:48.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/14/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1e8be9f : FAILURE in 47m 21s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/14/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/0c10d92 : FAILURE in 1h 02m 15s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":14},{"id":"ec97a109da5808aaf828bdeda773303ec6ea8c7f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-30 18:30:45.000000000","message":"Patch Set 14:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/14/check-cinder/Cinder-ISER-ISCSI/d190414 : FAILURE in 53m 12s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/14/check-cinder/Cinder-ISER-LIO/55cacc7 : FAILURE in 1h 00m 19s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":14},{"id":"0e9ac3a23c353fc79427420896ddc0235321c6d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-30 19:55:14.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/14/check/openstack-tox-lower-constraints/c07a32f/ : FAILURE in 11m 55s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/14/check/openstack-tox-pep8/d98296c/ : SUCCESS in 9m 01s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/14/check/openstack-tox-py27/2c83aa7/ : FAILURE in 9m 34s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/14/check/openstack-tox-py35/1d30fe4/ : FAILURE in 9m 49s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/14/check/openstack-tox-py36/2f05b97/ : FAILURE in 9m 13s\n- openstack-tox-docs http://logs.openstack.org/85/570885/14/check/openstack-tox-docs/f7830e4/html/ : SUCCESS in 9m 58s\n- tempest-full http://logs.openstack.org/85/570885/14/check/tempest-full/05f44a2/ : FAILURE in 1h 26m 36s\n- neutron-grenade http://logs.openstack.org/85/570885/14/check/neutron-grenade/07ff4f1/ : FAILURE in 57m 11s\n- grenade-py3 http://logs.openstack.org/85/570885/14/check/grenade-py3/1a6bc9f/ : FAILURE in 1h 00m 49s\n- tempest-full-py3 http://logs.openstack.org/85/570885/14/check/tempest-full-py3/86979c5/ : FAILURE in 1h 08m 53s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/14/check/cinder-tempest-dsvm-lvm-lio-barbican/88cf974/ : FAILURE in 2h 26m 16s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/14/check/cinder-tox-bandit-baseline/ad5a242/ : SUCCESS in 7m 09s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/14/check/nova-multiattach/5add45f/ : FAILURE in 1h 23m 45s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/14/check/openstack-tox-functional-py35/df49721/ : FAILURE in 5m 30s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/14/check/openstack-tox-functional-py36/26e0615/ : FAILURE in 7m 12s\n- cinder-rally-task http://logs.openstack.org/85/570885/14/check/cinder-rally-task/5549a6f/ : FAILURE in 36m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/14/check/openstack-tox-pylint/ee5e12a/ : FAILURE in 5m 35s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/14/check/devstack-plugin-ceph-tempest/b1f7070/ : TIMED_OUT in 2h 04m 49s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/14/check/openstack-tox-functional/222dd9d/ : FAILURE in 7m 05s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/14/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/b6d57c6/ : SUCCESS in 1h 07m 37s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/14/check/legacy-tempest-dsvm-lvm-multibackend/c7a2b5b/ : FAILURE in 2h 02m 39s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/14/check/legacy-tempest-dsvm-full-drbd-devstack/13c17f3/ : FAILURE in 1h 28m 16s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/14/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2868a56/ : FAILURE in 1h 32m 22s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/14/check/tempest-slow/d77cc71/ : FAILURE in 2h 30m 28s","accounts_in_message":[],"_revision_number":14},{"id":"7735a59b1b39b436ce25f6c272b2b89f755988cd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-01-31 06:35:29.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/85/570885/14/check/EMC_Unity_FC/5b0e32e/EMC_Unity_FC/970 : FAILURE in 47m 21s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/14/check/EMC_Unity_iSCSI/045a2b8/EMC_Unity_iSCSI/927 : FAILURE in 1h 43m 41s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"c5b2bc17c7b06c04620f729f2894c2e2038dc866","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-01-31 07:14:27.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-storpoolci\u0027.  For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/StorPool_CI\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/14/check/dsvm-tempest-storpool/bae5bbd : FAILURE in 1h 39m 56s","accounts_in_message":[],"_revision_number":14},{"id":"6d6c589a439c328d61cd5294d38a5c935f3ff37f","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-31 08:12:53.000000000","message":"Patch Set 14:\n\n@reviewers: The functionality works and i tested it for some time using tempest. However i have a very hard time mocking a snapshot in the corresponding test_remotefs unit tests, helpful hints are warmly welcome.","accounts_in_message":[],"_revision_number":14},{"id":"c1dd034ed55bee3e8a3eea854d0a9ab1e8f8357f","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-31 10:47:12.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"42cd19aadd3864e77c3d806f665732fd66db9595","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-31 10:48:03.000000000","message":"Patch Set 15:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/15/check/dsvm-veritas-access-cinder-driver/72f9c3a : FAILURE in 12s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/15/check/dsvm-veritas-access-cinder-iscsi-driver/5c76f4f : FAILURE in 12s","accounts_in_message":[],"_revision_number":15},{"id":"a374906faba6f2b594af45e8cb828faadaf4b116","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-31 10:54:15.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"93e7a14c2a0251903dfb50a61f3376664fc48056","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-31 10:55:39.000000000","message":"Patch Set 16:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/85/570885/16/check/dsvm-veritas-access-cinder-driver/954f5f6 : FAILURE in 11s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/85/570885/16/check/dsvm-veritas-access-cinder-iscsi-driver/f3f561e : FAILURE in 11s","accounts_in_message":[],"_revision_number":16},{"id":"2b6ebafd1d42459dc197e1a861767b75a3a1e8d1","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-01-31 10:59:51.000000000","message":"Patch Set 16:\n\nNote: Last updates removed context as a method argument from create_cloned_volume as that change has unwanted impact on other drivers. Context is now passed as an attribute of the source volume object.\n\nThe snapshot mocking issue persist, thought.","accounts_in_message":[],"_revision_number":16},{"id":"80c85c426df399afdf803620a4356e9bf81d98c6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-31 11:05:25.000000000","message":"Patch Set 15:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/15 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"88feb2098b4234ab958970b02991305a7e1eee43","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-31 11:09:32.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/16/iscsi/ : FAILURE in 14m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/16/smb/ : FAILURE in 13m 26s","accounts_in_message":[],"_revision_number":16},{"id":"f6d0590a97daecb06698830840d9f0110b0b45a4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-31 11:27:40.000000000","message":"Patch Set 16:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"73bcc6a09d3fbc6f3debe61ba4969661456b4ec7","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-01-31 11:34:52.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/730 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1067 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/906 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"11d7dd8bfb7a9a4d53a1486905560ed6468c5795","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-01-31 11:55:40.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/731 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1068 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/907 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"64ed3abc7992be3ea4de330d5994becdf9efdcd6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-31 11:55:41.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/16/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/0173b7f : SUCCESS in 56m 53s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/16/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/397d67e : FAILURE in 44s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":16},{"id":"10814e050af69a2cac0c0c68675af44739ee9cc5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-31 11:58:33.000000000","message":"Patch Set 16:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/16/check-not-vote/ext-cinder-zuul/2ff7840 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":16},{"id":"b14032753e902f63386fbfa636dcf67c0b78cbc4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-31 12:09:45.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/16/check/tempest-dsvm-full-xenial/96c1094/ : SUCCESS in 1h 08m 18s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/16/check/devstack-plugin-ceph-dsvm-xenial/c22a594/ : SUCCESS in 1h 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f415220f0cc8dd60646c3f95735f589cd7644a71","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-31 12:49:17.000000000","message":"Patch Set 16:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/16/check/fujitsu-eternus-dx-iscsi/e41d783 : SUCCESS in 1h 51m 31s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/16/check/fujitsu-eternus-dx-fc/71ce5a2 : SUCCESS in 1h 52m 26s","accounts_in_message":[],"_revision_number":16},{"id":"4dd2bf32e3608f490246b868e7add19671cd4fe0","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-31 12:53:38.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/16/check/dsvm-tempest-kvm/9e9833e : FAILURE in 39m 40s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/16/check/dsvm-tempest-exe/a334f07 : SUCCESS in 51m 23s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"a4fc6f903aca959c3a9eee07d0ef3ce6584c79e5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-31 13:04:45.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/16/check/dsvm-tempest-cisco-zonemanager-job/03534bf/ : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":16},{"id":"11342d0ff3df6249b4c42d110b9deae2736de0e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 13:26:54.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/16/check/openstack-tox-lower-constraints/be578d3/ : FAILURE in 10m 25s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/16/check/openstack-tox-pep8/4747582/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/16/check/openstack-tox-py27/2185f75/ : SUCCESS in 9m 48s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/16/check/openstack-tox-py35/ec1f43d/ : SUCCESS in 9m 55s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/16/check/openstack-tox-py36/94e27c3/ : FAILURE in 9m 31s\n- openstack-tox-docs http://logs.openstack.org/85/570885/16/check/openstack-tox-docs/bba4cd5/html/ : SUCCESS in 9m 12s\n- tempest-full http://logs.openstack.org/85/570885/16/check/tempest-full/291dbd5/ : SUCCESS in 1h 54m 40s\n- neutron-grenade http://logs.openstack.org/85/570885/16/check/neutron-grenade/82d31dc/ : SUCCESS in 1h 02m 21s\n- grenade-py3 http://logs.openstack.org/85/570885/16/check/grenade-py3/ad031d6/ : SUCCESS in 1h 04m 21s\n- tempest-full-py3 http://logs.openstack.org/85/570885/16/check/tempest-full-py3/2207d69/ : SUCCESS in 1h 29m 49s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/16/check/cinder-tempest-dsvm-lvm-lio-barbican/2336517/ : SUCCESS in 1h 49m 51s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/16/check/cinder-tox-bandit-baseline/a2f2d64/ : SUCCESS in 9m 03s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/16/check/nova-multiattach/e45dca8/ : SUCCESS in 1h 24m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/16/check/openstack-tox-functional-py35/7b241af/ : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/16/check/openstack-tox-functional-py36/d949b3e/ : SUCCESS in 6m 28s\n- cinder-rally-task http://logs.openstack.org/85/570885/16/check/cinder-rally-task/cb2800c/ : SUCCESS in 38m 00s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/16/check/openstack-tox-pylint/ecd9a77/ : FAILURE in 5m 33s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/16/check/devstack-plugin-ceph-tempest/322b6ee/ : SUCCESS in 1h 39m 09s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/16/check/openstack-tox-functional/fba2623/ : SUCCESS in 6m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/16/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/cae65d3/ : SUCCESS in 1h 02m 23s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/16/check/legacy-tempest-dsvm-lvm-multibackend/c2dfbb9/ : SUCCESS in 1h 45m 33s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/16/check/legacy-tempest-dsvm-full-drbd-devstack/03fc0ca/ : SUCCESS in 1h 29m 14s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/16/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/c8a8357/ : FAILURE in 1h 25m 29s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/16/check/tempest-slow/31c1077/ : SUCCESS in 2h 04m 55s","accounts_in_message":[],"_revision_number":16},{"id":"d66d88d32d5f9c21f5e887875350569c4b7808df","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-31 15:01:51.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/16/check/dsvm-tempest-infinibox-fc/42c3e85 : SUCCESS in 59m 05s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/16/check/dsvm-tempest-infinibox-iscsi/c3a7fab : SUCCESS in 1h 25m 17s","accounts_in_message":[],"_revision_number":16},{"id":"edbc36f6f402497d9cacebc19c35eb275bbdd9ee","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-31 15:58:57.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/16/check/Cinder-AS13000/45afa2f : SUCCESS in 2h 45m 34s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/16/check/cinder-test-fc/38273ab : SUCCESS in 1h 06m 23s\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/16/check/cinder-test-iscsi/2dde5b3 : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":16},{"id":"deb19a9903fd407d1dfbe2b7c7af53f802f96854","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-02-01 00:26:25.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/85/570885/16/check/EMC_Unity_FC/d03eca1/EMC_Unity_FC/982 : FAILURE in 1h 27m 20s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/16/check/EMC_Unity_iSCSI/b69e8b5/EMC_Unity_iSCSI/938 : FAILURE in 1h 31m 10s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"152ea3d0f55da07a84f4daca23e139d6dc21fcf9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-02-01 02:54:17.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/16/check/dsvm-tempest-storpool/218879c : SUCCESS in 1h 46m 25s","accounts_in_message":[],"_revision_number":16},{"id":"8d354a04ba70823ca03b7f1c733fbb3beb9b4774","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-02-01 08:58:54.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/16/check/NEC-ISCSI/7e1acbe : SUCCESS in 1h 32m 35s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/16/check/NEC-FC/5bfffb6 : SUCCESS in 1h 35m 49s","accounts_in_message":[],"_revision_number":16},{"id":"f6af0466e77e497de06fc0c2350c27ae492504da","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-02-01 14:10:36.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"22ba543737c7dbd225a9bce80ead1c76484c8c0a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-01 14:48:44.000000000","message":"Patch Set 16:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/16/check-not-vote/ext-cinder-zuul/333401f : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":16},{"id":"fa60071b425836abda83e86dd68a56c0b90ade56","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-02-01 15:07:46.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/16/check/dsvm-tempest-kvm/781cc81 : FAILURE in 56m 37s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/16/check/dsvm-tempest-exe/2a95105 : SUCCESS in 49m 11s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"999e4ea3acd5958118ff8b4d48fe01469798cbc5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-01 15:27:59.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/16/check/tempest-dsvm-full-xenial/a0780e0/ : SUCCESS in 1h 14m 39s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/16/check/devstack-plugin-ceph-dsvm-xenial/d76ba5c/ : SUCCESS in 1h 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"90d139450e15cb3c2e825b68358b9980952fcfc7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-02-01 15:32:34.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/16/check/dsvm-tempest-cisco-zonemanager-job/6692b33/ : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":16},{"id":"9f6335f357359dcd25d81d1e9d907b8b37fc6104","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-02-01 16:00:36.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/16/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c290710 : SUCCESS in 53m 58s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/16/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/b36fae1 : FAILURE in 1h 49m 44s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":16},{"id":"5d840eac55841dbf734ac2d7e5df892111a3a0da","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-02-01 16:28:41.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/973 : FAILURE (Comment recheck linstor to rerun the CI)","accounts_in_message":[],"_revision_number":16},{"id":"c7e3a46a20a7859c1aed7834daafe635a8b8a7c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-01 16:51:16.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/16/check/openstack-tox-lower-constraints/b475bf5/ : SUCCESS in 12m 33s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/16/check/openstack-tox-pep8/a52755c/ : SUCCESS in 8m 08s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/16/check/openstack-tox-py27/b27c438/ : FAILURE in 9m 52s\n- openstack-tox-py35 http://logs.openstack.org/85/570885/16/check/openstack-tox-py35/6719f9e/ : SUCCESS in 8m 54s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/16/check/openstack-tox-py36/cd5ef67/ : SUCCESS in 9m 46s\n- openstack-tox-docs http://logs.openstack.org/85/570885/16/check/openstack-tox-docs/6b1cf63/html/ : SUCCESS in 8m 11s\n- tempest-full http://logs.openstack.org/85/570885/16/check/tempest-full/26a66fa/ : SUCCESS in 1h 47m 37s\n- neutron-grenade http://logs.openstack.org/85/570885/16/check/neutron-grenade/96a0a4c/ : SUCCESS in 1h 01m 58s\n- grenade-py3 http://logs.openstack.org/85/570885/16/check/grenade-py3/398475a/ : SUCCESS in 1h 00m 44s\n- tempest-full-py3 http://logs.openstack.org/85/570885/16/check/tempest-full-py3/5b75a33/ : SUCCESS in 1h 27m 23s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/16/check/cinder-tempest-dsvm-lvm-lio-barbican/e468670/ : SUCCESS in 2h 15m 30s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/16/check/cinder-tox-bandit-baseline/87086b9/ : SUCCESS in 8m 17s (non-voting)\n- nova-multiattach http://logs.openstack.org/85/570885/16/check/nova-multiattach/cfe3d20/ : SUCCESS in 1h 23m 45s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/570885/16/check/openstack-tox-functional-py35/6347be5/ : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/16/check/openstack-tox-functional-py36/868150d/ : SUCCESS in 5m 12s\n- cinder-rally-task http://logs.openstack.org/85/570885/16/check/cinder-rally-task/88247fb/ : SUCCESS in 43m 45s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/16/check/openstack-tox-pylint/7738322/ : FAILURE in 4m 28s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/16/check/devstack-plugin-ceph-tempest/de2be90/ : SUCCESS in 1h 57m 26s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/16/check/openstack-tox-functional/d945a2a/ : SUCCESS in 6m 49s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/16/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/25677f4/ : SUCCESS in 1h 16m 21s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/16/check/legacy-tempest-dsvm-lvm-multibackend/733b161/ : SUCCESS in 1h 39m 26s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/85/570885/16/check/legacy-tempest-dsvm-full-drbd-devstack/a7e88e8/ : SUCCESS in 1h 32m 18s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/16/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/c5e995c/ : FAILURE in 1h 12m 19s (non-voting)\n- tempest-slow http://logs.openstack.org/85/570885/16/check/tempest-slow/2b4abf2/ : SUCCESS in 2h 32m 55s","accounts_in_message":[],"_revision_number":16},{"id":"985154688a03d95b768044ceb9a1d6c78730de86","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-02-01 20:02:23.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/16/check/Cinder-AS13000/772057e : SUCCESS in 2h 30m 55s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/16/check/cinder-test-fc/1933293 : SUCCESS in 56m 57s\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/16/check/cinder-test-iscsi/b63174c : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":16},{"id":"d66fd239f77cea6cefeb50dc92b492f48e42b208","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-02-02 01:36:26.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/85/570885/16/check/EMC_XtremIO_FC/d636ca7/EMC_XtremIO_FC/1434 : SUCCESS in 1h 31m 29s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/16/check/EMC_XtremIO_iSCSI/d5140e4/EMC_XtremIO_iSCSI/1464 : SUCCESS in 1h 21m 02s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":16},{"id":"bc39fc7ca4452280715fd23c1ed82ce5580047c1","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-02-02 09:34:07.000000000","message":"Patch Set 16:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/988 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"2ea3550f85de0683b0bb4e33141a3d66509334b3","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-02-02 19:52:49.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12784 : FAILURE in 35m 10s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9142 : SUCCESS in 1h 00m 19s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10894 : FAILURE in 1h 08m 02s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9521 : FAILURE in 1h 50m 25s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10490 : FAILURE in 37m 27s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1433 : SUCCESS in 1h 53m 46s","accounts_in_message":[],"_revision_number":16},{"id":"cfeb62784526b4fb9a49fe6bf3d08fa80e19a6cd","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-05 13:10:48.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"7be7513fff6d4af2502f1ffe17246596d0615f9a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-05 13:52:09.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/17/check-not-vote/ext-cinder-zuul/e756235 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":17},{"id":"e3b601d80375bf597f28cce4711569a037733144","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-05 13:59:26.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/17/check/dsvm-tempest-kvm/a22c273 : FAILURE in 46m 45s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/17/check/dsvm-tempest-exe/44d58ad : FAILURE in 47m 24s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"ac06acaad8b85f6ad318ee3bcba561d6b36cbbad","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-04-05 14:02:44.000000000","message":"Patch Set 17:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-04-05_06-10-54/ : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"38ef89f091ca102c621caadddca6a80fd16148c8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-04-05 14:11:17.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"07f33dd5d550ee38f586b3fa8c13a0682456b4bb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-05 14:21:01.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/17/check/dsvm-tempest-infinibox-fc/986887e : SUCCESS in 58m 16s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/17/check/dsvm-tempest-infinibox-iscsi/af805bf : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":17},{"id":"860c89014cd537538657d25ffbf9205e9f59bbd2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-05 14:21:02.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/17/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/fa1248a : FAILURE in 46m 05s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/17/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/5128be2 : SUCCESS in 1h 09m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":17},{"id":"06de7d10fb2527111514f0e97b551095b417f7fd","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-05 14:36:46.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/17/iscsi/ : FAILURE in 1h 25m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/17/smb/ : FAILURE in 50m 29s","accounts_in_message":[],"_revision_number":17},{"id":"3c2e4dbe65069c1b2821ed727dc6f5fcf3da28a9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-04-05 14:41:54.000000000","message":"Patch Set 17:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/17 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"914f318e13e6e6aa843115541f6abe897aeecf38","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-04-05 14:44:58.000000000","message":"Patch Set 17:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/17/check/fujitsu-eternus-dx-iscsi/07e5b04 : SUCCESS in 1h 33m 30s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/17/check/fujitsu-eternus-dx-fc/35aac25 : SUCCESS in 1h 31m 45s","accounts_in_message":[],"_revision_number":17},{"id":"92855b2e91c7e19df2dfc70df128dcd56260de3f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-04-05 14:47:37.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/85/570885/17/check/EMC_Unity_FC/9889009/EMC_Unity_FC/1400 : SUCCESS in 1h 25m 00s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/17/check/EMC_Unity_iSCSI/c8dc2ed/EMC_Unity_iSCSI/1333 : FAILURE in 1h 36m 34s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"4bc6cf7aaa23dd90aa899ed839f325f6bdbd7fb3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-04-05 15:00:58.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/85/570885/17/check/dsvm-tempest-storpool/69c40d1 : SUCCESS in 1h 49m 35s","accounts_in_message":[],"_revision_number":17},{"id":"475f108b3667bf3301d6eb2246899e4809bc51f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-05 15:06:14.000000000","message":"Patch Set 17:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/17/check/tempest-dsvm-full-xenial/76bcef5/ : SUCCESS in 1h 20m 37s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/17/check/devstack-plugin-ceph-dsvm-xenial/d2f4812/ : SUCCESS in 1h 55m 02s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"fe8ad299a2c0be15159b2a35396f916b0800d063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 15:17:59.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/17/check/openstack-tox-lower-constraints/eeb5743/ : SUCCESS in 10m 22s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/17/check/openstack-tox-pep8/67968b6/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/17/check/openstack-tox-py27/89ecb03/ : SUCCESS in 8m 50s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/17/check/openstack-tox-py36/e9131ef/ : SUCCESS in 7m 55s\n- openstack-tox-docs http://logs.openstack.org/85/570885/17/check/openstack-tox-docs/ed4142e/html/ : SUCCESS in 8m 09s\n- tempest-full http://logs.openstack.org/85/570885/17/check/tempest-full/17a3113/ : SUCCESS in 1h 33m 46s\n- neutron-grenade http://logs.openstack.org/85/570885/17/check/neutron-grenade/d852623/ : SUCCESS in 54m 36s (non-voting)\n- grenade-py3 http://logs.openstack.org/85/570885/17/check/grenade-py3/13ea2e6/ : SUCCESS in 1h 02m 04s\n- tempest-full-py3 http://logs.openstack.org/85/570885/17/check/tempest-full-py3/a903f03/ : SUCCESS in 1h 21m 25s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/17/check/cinder-tempest-dsvm-lvm-lio-barbican/4accc10/ : FAILURE in 1h 18m 44s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/17/check/cinder-tox-bandit-baseline/a70acff/ : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/17/check/openstack-tox-functional-py36/329bd16/ : SUCCESS in 5m 16s\n- cinder-rally-task http://logs.openstack.org/85/570885/17/check/cinder-rally-task/ec7c48c/ : SUCCESS in 33m 55s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/17/check/openstack-tox-pylint/8c0d2b1/ : FAILURE in 3m 56s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/17/check/devstack-plugin-ceph-tempest/df57ea5/ : SUCCESS in 1h 43m 57s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/17/check/openstack-tox-functional/8b568ae/ : SUCCESS in 5m 23s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/17/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/d634fbe/ : SUCCESS in 56m 06s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/17/check/legacy-tempest-dsvm-lvm-multibackend/4ae76e9/ : SUCCESS in 1h 13m 16s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/17/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2e635b7/ : FAILURE in 1h 11m 01s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/17/check/tempest-slow-py3/59edcb5/ : SUCCESS in 2h 04m 40s","accounts_in_message":[],"_revision_number":17},{"id":"c92bad786bee8743b1a72b539c94f9074de8b536","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-04-05 15:55:20.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/17/check/Cinder-AS13000/d9bc8e9 : SUCCESS in 2h 30m 19s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/17/check/cinder-test-fc/08f478b : SUCCESS in 1h 14m 43s\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/17/check/cinder-test-iscsi/9fb2c9f : SUCCESS in 1h 21m 49s","accounts_in_message":[],"_revision_number":17},{"id":"35e0b186463cb607b05c11e8ddbc2d882ee56e0a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-05 16:47:48.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  To recheck use \u0027recheck\u0027.\n\n- dsvm-tempest-cisco-zonemanager-job finger://localhost.localdomain:17979/b1b8ccc126df44ff94f26a968927a0bc : RETRY_LIMIT in 1h 01m 15s","accounts_in_message":[],"_revision_number":17},{"id":"25ad8b12c5a98e89a491b5582429474abca6c8c9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-04-05 16:55:58.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/17/check-cinder/Cinder-ISER-ISCSI/8be608d : SUCCESS in 1h 05m 38s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/17/check-cinder/Cinder-ISER-LIO/786c9d4 : SUCCESS in 1h 03m 51s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":17},{"id":"dc0a8448baae240bbaff888e1b49b11f1cf2b89a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-04-05 17:42:56.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/17/check/NEC-ISCSI/aac7cd2 : SUCCESS in 2h 57m 43s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/17/check/NEC-FC/baa83aa : FAILURE in 2h 59m 15s","accounts_in_message":[],"_revision_number":17},{"id":"a1168b8f94a40b76566e78fffe88ab7a8f9f43af","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-12 10:37:14.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9cb2d2a57a14ea507544f838b0cc4cb0f0e2ab06","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-12 10:39:02.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"317c36d53af5a8cd96332a050a37c15bb07ffb59","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-12 10:52:26.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/18/iscsi/ : FAILURE in 14m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/18/smb/ : FAILURE in 14m 43s","accounts_in_message":[],"_revision_number":18},{"id":"c916d3d4fa07e1d4f9c2975a89d8c38c4763363f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-04-12 12:08:41.000000000","message":"Patch Set 18:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/18/check/fujitsu-eternus-dx-iscsi/896ba27 : FAILURE in 16m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/18/check/fujitsu-eternus-dx-fc/35dd1a6 : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":18},{"id":"8e2a21f73c5f85a681a238e576dd89071a1c660f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 12:19:16.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/18/check/dsvm-tempest-kvm/bc4174e : FAILURE in 53m 04s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/18/check/dsvm-tempest-exe/e6aa213 : FAILURE in 43m 17s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"e594a7a9e22cd3ae5e242ea92b588062e32e5058","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-04-12 12:25:11.000000000","message":"Patch Set 18:\n\nBuild successful. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-04-12_04-34-46/ : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"bc51880d3288a2d611893ebb27e0ea00ce7b42ed","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-12 12:31:04.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/18/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/bd5fb2a : FAILURE in 1h 51m 53s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/18/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/6a9f12c : SUCCESS in 1h 03m 25s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":18},{"id":"4f4571bcc53a7fb7c584c7d1bbb72a361189b2b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 12:46:49.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/18/check/openstack-tox-lower-constraints/1978382/ : SUCCESS in 14m 30s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/18/check/openstack-tox-pep8/2d8f168/ : SUCCESS in 9m 25s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/18/check/openstack-tox-py27/29a3397/ : SUCCESS in 12m 57s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/18/check/openstack-tox-py36/08f9962/ : SUCCESS in 12m 11s\n- openstack-tox-docs http://logs.openstack.org/85/570885/18/check/openstack-tox-docs/dce0f41/html/ : SUCCESS in 8m 13s\n- tempest-full http://logs.openstack.org/85/570885/18/check/tempest-full/ea01b90/ : SUCCESS in 2h 03m 04s\n- neutron-grenade http://logs.openstack.org/85/570885/18/check/neutron-grenade/f2e4c7b/ : SUCCESS in 57m 07s\n- grenade-py3 http://logs.openstack.org/85/570885/18/check/grenade-py3/b359bac/ : SUCCESS in 1h 00m 38s\n- tempest-full-py3 http://logs.openstack.org/85/570885/18/check/tempest-full-py3/a012dfe/ : SUCCESS in 1h 44m 41s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/18/check/cinder-tempest-dsvm-lvm-lio-barbican/7c7f4cd/ : FAILURE in 1h 20m 49s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/18/check/cinder-tox-bandit-baseline/9cb67d0/ : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/18/check/openstack-tox-functional-py36/bd2993e/ : SUCCESS in 8m 10s\n- cinder-rally-task http://logs.openstack.org/85/570885/18/check/cinder-rally-task/06ca26a/ : SUCCESS in 44m 48s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/18/check/openstack-tox-pylint/346254d/ : SUCCESS in 7m 03s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/18/check/devstack-plugin-ceph-tempest/ca82fbd/ : SUCCESS in 1h 25m 00s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/18/check/openstack-tox-functional/88de545/ : POST_FAILURE in 15m 01s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/18/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/f2647c9/ : FAILURE in 46m 13s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/18/check/legacy-tempest-dsvm-lvm-multibackend/8f85076/ : SUCCESS in 1h 40m 25s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/18/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/15c3d64/ : FAILURE in 1h 21m 00s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/18/check/tempest-slow-py3/3075138/ : SUCCESS in 2h 06m 09s","accounts_in_message":[],"_revision_number":18},{"id":"9954582bbb8a8d525393821a97daba68f0a7d34c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-12 13:00:25.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/18/check/dsvm-tempest-cisco-zonemanager-job/bda46a7/ : FAILURE in 1h 01m 01s","accounts_in_message":[],"_revision_number":18},{"id":"6635e0d6ecc68dd9a879f4b704526b4833669d73","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-12 13:20:59.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"b3a166af2d517400c9b1eca5d1c534e2eca0e856","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-04-12 13:48:46.000000000","message":"Patch Set 18:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"feac7088d3ac0ec4c29e231c1f096d8e50ace397","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 14:08:28.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/18/check/dsvm-tempest-kvm/99a64ee : FAILURE in 46m 18s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/18/check/dsvm-tempest-exe/ab445c2 : FAILURE in 46m 15s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"304ef54f6aacb3d738f6012ac426434245e079fb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-12 14:32:57.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/18/check/dsvm-tempest-cisco-zonemanager-job/ff3442b/ : FAILURE in 1h 00m 39s","accounts_in_message":[],"_revision_number":18},{"id":"fd230ac079a7904a79a893bcd60e3b7a31c350ac","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-12 14:34:45.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/18/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/267fc8b : SUCCESS in 54m 32s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/18/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/7e46630 : SUCCESS in 1h 13m 28s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":18},{"id":"38c32962e378e94321b77821c5c6a88465afee3c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-12 14:43:25.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/18/iscsi/ : FAILURE in 1h 21m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/18/smb/ : FAILURE in 51m 18s","accounts_in_message":[],"_revision_number":18},{"id":"a2a43124e8bf0248e959b5ecaad30573657480b4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-12 15:25:43.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/18/check/dsvm-tempest-infinibox-fc/e6d190e : SUCCESS in 1h 07m 52s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/18/check/dsvm-tempest-infinibox-iscsi/5331524 : SUCCESS in 1h 15m 09s","accounts_in_message":[],"_revision_number":18},{"id":"52aa63d9151c7add9030c212d32141d19db351b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 15:26:52.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/18/check/openstack-tox-lower-constraints/65d7e8e/ : SUCCESS in 14m 35s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/18/check/openstack-tox-pep8/533a438/ : SUCCESS in 8m 33s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/18/check/openstack-tox-py27/09163f8/ : SUCCESS in 10m 47s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/18/check/openstack-tox-py36/642829c/ : SUCCESS in 8m 20s\n- openstack-tox-docs http://logs.openstack.org/85/570885/18/check/openstack-tox-docs/5da6e83/html/ : SUCCESS in 8m 56s\n- tempest-full http://logs.openstack.org/85/570885/18/check/tempest-full/dac8eb0/ : SUCCESS in 1h 31m 32s\n- neutron-grenade http://logs.openstack.org/85/570885/18/check/neutron-grenade/01e3bac/ : SUCCESS in 54m 03s\n- grenade-py3 http://logs.openstack.org/85/570885/18/check/grenade-py3/8b63471/ : SUCCESS in 54m 32s\n- tempest-full-py3 http://logs.openstack.org/85/570885/18/check/tempest-full-py3/d153044/ : SUCCESS in 1h 26m 55s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/18/check/cinder-tempest-dsvm-lvm-lio-barbican/a6fc9da/ : SUCCESS in 1h 34m 27s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/18/check/cinder-tox-bandit-baseline/98cf2cb/ : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/18/check/openstack-tox-functional-py36/cd193b1/ : SUCCESS in 5m 44s\n- cinder-rally-task http://logs.openstack.org/85/570885/18/check/cinder-rally-task/3705dd1/ : SUCCESS in 36m 32s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/18/check/openstack-tox-pylint/eaefe00/ : SUCCESS in 5m 20s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/18/check/devstack-plugin-ceph-tempest/ffe10dc/ : SUCCESS in 1h 26m 47s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/18/check/openstack-tox-functional/fe8c2cd/ : SUCCESS in 5m 38s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/18/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/a3efe57/ : SUCCESS in 1h 02m 34s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/18/check/legacy-tempest-dsvm-lvm-multibackend/0bdb53e/ : SUCCESS in 1h 19m 52s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/18/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/b6d6c80/ : FAILURE in 1h 41m 49s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/18/check/tempest-slow-py3/6349310/ : SUCCESS in 2h 03m 22s","accounts_in_message":[],"_revision_number":18},{"id":"aaba9738231adffa15f3e93658367e24914e71f4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 15:33:43.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS http://publiclogs.emc.com/85/570885/18/check/EMC_VxFlexOS/e56730b/EMC_VxFlexOS/101 : FAILURE in 6m 33s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":18},{"id":"1b20eabf0ef654d8ef023b20f55e91328dd4125f","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-04-12 15:57:59.000000000","message":"Patch Set 18:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-85-570885-18/nedge_iscsi : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":18},{"id":"bd39dd996419dbe5847afa8e284e6866ff3ed00c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-04-12 16:05:37.000000000","message":"Patch Set 18:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-85-570885-18/ns5_iscsi : SUCCESS in 37m 43s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-85-570885-18/ns5_nfs : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":18},{"id":"8f5eec93e18a3ca488111f2658c99eeea3edbd4d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-12 16:21:47.000000000","message":"Patch Set 18:\n\nBuild failed\n\n- ext-cinder-zuul : EXCEPTION","accounts_in_message":[],"_revision_number":18},{"id":"bdd91417402004b7a6d4555116b44785325bfe1a","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-04-12 16:23:15.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/85/570885/18/check/synology-dsm-cinder-iscsi/804efcc : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":18},{"id":"a73df8d09eb2b0a369c97b6c99c0b001d2fa9051","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-12 16:42:21.000000000","message":"Patch Set 18:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"e17a75a48bce68b808c0ed1b9f72e5599690ad15","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-04-12 16:44:25.000000000","message":"Patch Set 18:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-85-570885-18 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI","accounts_in_message":[],"_revision_number":18},{"id":"06e5ca9e9409938c3c41d77f7c59f1930fb91f76","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-04-12 16:44:46.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Dell_SC_FC http://publiclogs.emc.com/85/570885/18/check/Dell_SC_FC/c2e39f4/Dell_SC_FC/1028 : FAILURE in 1h 17m 26s\n- Dell_SC_iSCSI http://publiclogs.emc.com/85/570885/18/check/Dell_SC_iSCSI/cee1887/Dell_SC_iSCSI/1013 : FAILURE in 6s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":18},{"id":"b92e87c5ec439dc3fd6262d67fbfe0e492a9c8bc","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 16:47:39.000000000","message":"Patch Set 18:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-85-570885-18-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"3ba080c7daa79340c57964baec493cadef87f1fd","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 17:31:45.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VNX_iSCSI http://publiclogs.emc.com/85/570885/18/check/EMC_VNX_iSCSI/9724cfe/EMC_VNX_iSCSI/636 : FAILURE in 6s\n- EMC_VNX_FC http://publiclogs.emc.com/85/570885/18/check/EMC_VNX_FC/d1ce52b/EMC_VNX_FC/651 : SUCCESS in 2h 04m 36s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":18},{"id":"43a051dcb63eda2c72ec62bb7c3363616831ffae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-04-12 17:47:53.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/18/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ad3e0de : SUCCESS in 2h 05m 58s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/18/check/3par-fc-driver-master-client-pip-ssa02-dsvm/b8192e4 : SUCCESS in 2h 18m 23s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/85/570885/18/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/98f2c50 : SUCCESS in 1h 59m 22s","accounts_in_message":[],"_revision_number":18},{"id":"2258ead7c3d2808512a3fffb69f5bebf9e3ae0d4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 17:54:21.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/85/570885/18/check/EMC_XtremIO_FC/52ffb85/EMC_XtremIO_FC/2125 : SUCCESS in 1h 06m 47s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/18/check/EMC_XtremIO_iSCSI/b7e1a59/EMC_XtremIO_iSCSI/2102 : SUCCESS in 1h 21m 45s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":18},{"id":"85a215528cbb335969f2cbbe8a8671ec3c011e5e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-04-12 18:08:36.000000000","message":"Patch Set 18:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/85/570885/18/HPMSA-FC : SUCCESS in 56:36.87\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/85/570885/18/HPMSA-iSCSI : SUCCESS in 53:50.38","accounts_in_message":[],"_revision_number":18},{"id":"78508b5f06d88d9db3085eb001c95050917fff3d","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-04-12 19:21:29.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/85/570885/18/check/kaminario-dsvm-tempest-full-iscsi/9da61dd : SUCCESS in 2h 01m 42s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/85/570885/18/check/kaminario-dsvm-tempest-full-FC/3b901ca : SUCCESS in 1h 44m 26s","accounts_in_message":[],"_revision_number":18},{"id":"fe88a4fe09a0599764a06de868e198d7e4749e52","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 20:11:40.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/18/check/NEC-ISCSI/a3eaa60 : SUCCESS in 2h 10m 26s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/18/check/NEC-FC/ae63b39 : SUCCESS in 1h 59m 44s","accounts_in_message":[],"_revision_number":18},{"id":"4dd8ba69574c31dfb4ac83f88e8a13832de8683a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-12 22:09:49.000000000","message":"Patch Set 18:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/18/check/tempest-dsvm-full-xenial/c72e4d9/ : SUCCESS in 1h 05m 48s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/18/check/devstack-plugin-ceph-dsvm-xenial/cbb5946/ : FAILURE in 1h 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6fd41ea55f5d8cef44459e930775a936cb4d5e91","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-04-13 01:35:13.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/13261 : FAILURE in 1h 18m 09s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9629 : FAILURE in 36m 23s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/11349 : SUCCESS in 1h 07m 20s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/10003 : SUCCESS in 52m 38s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10975 : FAILURE in 30m 31s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1843 : FAILURE in 2m 24s","accounts_in_message":[],"_revision_number":18},{"id":"5e30df2b165bff4128fda58616ddf0a0c7e13821","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-04-13 06:18:43.000000000","message":"Patch Set 18:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/85/570885/18/Lenovo-FC : FAILURE in 5:51:51\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/85/570885/18/Lenovo-iSCSI : FAILURE in 4:45:23","accounts_in_message":[],"_revision_number":18},{"id":"539b64f7738c36dd828258caa9adeab29333131e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-04-13 07:55:13.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/85/570885/18/upstream-solidfire-check/cinder-solidfire-iSCSI/cacd648/ : NOT_REGISTERED\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":18},{"id":"27f01a1fe83e8b4daf213e0c0e50b9797ced4bf8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-04-14 17:24:01.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_FC http://publiclogs.emc.com/570885/18/EMC_VMAX_FC/1029 : FAILURE in 8h 09m 21s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/570885/18/EMC_VMAX_iSCSI/1061 : FAILURE in 7s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":18},{"id":"b45c031da11fc54e294410bda5774631906a1ffb","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-17 10:35:45.000000000","message":"Patch Set 18:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":18},{"id":"29119e66d28063d4600aced5358f0c3d6480274f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-17 11:30:39.000000000","message":"Patch Set 18:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"35010978e815ec554b19d503c1cba83fdfe52b9a","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-17 12:49:15.000000000","message":"Patch Set 18: Workflow-1\n\nfix this first","accounts_in_message":[],"_revision_number":18},{"id":"50c07c865983e98741b870ca1708eaceca25594d","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-18 14:31:50.000000000","message":"Patch Set 18:\n\nhmm.\n\u003d\u003d\u003d\u003d\u003d\u003d\nTotals\n\u003d\u003d\u003d\u003d\u003d\u003d\nRan: 249 tests in 1832.0000 sec.\n - Passed: 234\n - Skipped: 15\n - Expected Fail: 0\n - Unexpected Success: 0\n - Failed: 0\nSum of execute time for each test: 3513.2676 sec.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nWorker Balance\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n - Worker 0 (108 tests) \u003d\u003e 0:26:46.948679\n - Worker 1 (59 tests) \u003d\u003e 0:30:32.510749\n - Worker 2 (82 tests) \u003d\u003e 0:22:35.374405\n___________________________________ summary ____________________________________\n  all-plugin: commands succeeded\n  congratulations :)\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":18},{"id":"18373f2ab7042efa07db41ae244067e04dc637c9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-18 15:55:45.000000000","message":"Patch Set 18:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"bdbd0d93eff437e625164ca31c0578457b88cc39","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-18 16:05:10.000000000","message":"Patch Set 18:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":18},{"id":"a0f82214143225ca30e36dc01d6126a3de77fba2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-18 17:31:52.000000000","message":"Patch Set 18:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"2b19d499f8b04fd9d3e246dbb520eb79195d04b6","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-21 20:48:47.000000000","message":"Patch Set 18:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":18},{"id":"61e8ce47b226dea65e82e158547e0f52b3ae09a9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-21 21:04:24.000000000","message":"Patch Set 18:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"f367df9023fed19e20aa8b258fd2f6abbd1ec649","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-29 11:54:07.000000000","message":"Patch Set 18:\n\nrun-Quobyte CI\n\nTesting opendev domain migration","accounts_in_message":[],"_revision_number":18},{"id":"f7a160323b8a02b58f9138969fe1e710d08cfe2f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-29 12:07:04.000000000","message":"Patch Set 18:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"5c9b2647ab784d04ec10b1c6b71d9fd9ec7ef933","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-29 13:13:02.000000000","message":"Patch Set 18:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":18},{"id":"1b79d7d9c6d2e81731fa834cff3adbabd9a5e78b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-29 14:16:09.000000000","message":"Patch Set 18:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-18 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":18},{"id":"67e386df85ca664257b6c26a506d21aa735f3660","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-29 14:22:43.000000000","message":"Patch Set 18: -Workflow\n\nQuobyte CI has been migrated to opendev.org, tests ok now.","accounts_in_message":[],"_revision_number":18},{"id":"62b4f8b75e2bc7752bf2b652507eef66bf14831a","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-04-30 13:34:19.000000000","message":"Patch Set 18:\n\nrun-Virtuozzo Storage CI","accounts_in_message":[],"_revision_number":18},{"id":"0b2b411dcfee691688a877862a5fd7d0a3d221b1","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 13:35:46.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/18/check/dsvm-tempest-kvm/b6fec5e : FAILURE in 29s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/18/check/dsvm-tempest-exe/9acb5ef : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"f4ad27561cb86003add499b1fe1f2358b4de564a","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-05-02 06:24:42.000000000","message":"Patch Set 18:\n\nrun-Virtuozzo Storage CI","accounts_in_message":[],"_revision_number":18},{"id":"f674015a35a2dc74615c34f2f4cf12cf3ffe7606","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-02 06:26:13.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/18/check/dsvm-tempest-kvm/fb23434 : FAILURE in 3s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/18/check/dsvm-tempest-exe/a09fa62 : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"bf8845d95a2573c65c37884a97884ffb5547e086","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-05-24 08:49:26.000000000","message":"Patch Set 18:\n\nrecheck\n\nNeed NFS and vzstorage test logs for fixing those","accounts_in_message":[],"_revision_number":18},{"id":"98da19e28be0578ddfaa7e6b3f3bc4448c377ea0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-24 09:28:51.000000000","message":"Patch Set 18:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/18/check-not-vote/ext-cinder-zuul/3fbf01e : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":18},{"id":"18ee98b838fdf3606706b0c15f360a66785cfcae","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-24 09:40:31.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/18/iscsi/ : SUCCESS in 50m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/18/smb/ : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":18},{"id":"e776333f709571d7bee8af74e4d60e6454969413","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-24 10:18:12.000000000","message":"Patch Set 18:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/18/check/tempest-dsvm-full-xenial/554e218/ : SUCCESS in 1h 21m 32s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/18/check/devstack-plugin-ceph-dsvm-xenial/621d9bb/ : SUCCESS in 1h 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3fa86cacc20c87ca8788b6fbdd1237ed410296ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 11:21:59.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/18/check/openstack-tox-lower-constraints/722ce0d/ : SUCCESS in 13m 23s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/18/check/openstack-tox-pep8/47b73d9/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/18/check/openstack-tox-py27/ed5a396/ : SUCCESS in 9m 18s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/18/check/openstack-tox-py36/c498629/ : SUCCESS in 8m 33s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/18/check/openstack-tox-py37/c352db3/ : FAILURE in 15m 51s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/85/570885/18/check/openstack-tox-docs/c2ed736/html/ : SUCCESS in 7m 40s\n- tempest-full http://logs.openstack.org/85/570885/18/check/tempest-full/4ceee41/ : SUCCESS in 1h 33m 35s\n- neutron-grenade http://logs.openstack.org/85/570885/18/check/neutron-grenade/1d64f23/ : SUCCESS in 1h 03m 13s\n- grenade-py3 http://logs.openstack.org/85/570885/18/check/grenade-py3/dbfa8e1/ : SUCCESS in 1h 03m 05s\n- tempest-full-py3 http://logs.openstack.org/85/570885/18/check/tempest-full-py3/6aff886/ : FAILURE in 1h 19m 48s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/18/check/cinder-tempest-dsvm-lvm-lio-barbican/a50f1fc/ : SUCCESS in 1h 42m 05s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/18/check/cinder-tox-bandit-baseline/1b464ab/ : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/18/check/openstack-tox-functional-py36/4ead6a2/ : SUCCESS in 4m 44s\n- cinder-rally-task http://logs.openstack.org/85/570885/18/check/cinder-rally-task/9516811/ : SUCCESS in 42m 58s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/18/check/openstack-tox-pylint/3795755/ : SUCCESS in 4m 19s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/18/check/devstack-plugin-ceph-tempest/587e7e1/ : SUCCESS in 1h 30m 05s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/18/check/openstack-tox-functional/7837517/ : SUCCESS in 5m 00s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/18/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/8470a6d/ : SUCCESS in 1h 09m 56s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/18/check/legacy-tempest-dsvm-lvm-multibackend/9d464bc/ : SUCCESS in 1h 31m 28s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/18/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/f91f188/ : FAILURE in 1h 35m 34s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/18/check/tempest-slow-py3/bc8a2af/ : SUCCESS in 2h 11m 37s","accounts_in_message":[],"_revision_number":18},{"id":"da2712b6efcf5fa1de1e84f710bbdae080b0e5eb","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-05-24 14:46:41.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/85/570885/18/check/kaminario-dsvm-tempest-full-iscsi/b8965b8 : SUCCESS in 2h 47m 34s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/85/570885/18/check/kaminario-dsvm-tempest-full-FC/bee21d2 : SUCCESS in 2h 01m 13s","accounts_in_message":[],"_revision_number":18},{"id":"578127a118744fd3243ac35457baaf6fdb588190","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-24 22:31:09.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/85/570885/18/check/dsvm-tempest-zadara-driver/a8df530 : FAILURE in 1h 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6bc943a8ded14d8af13ee50b3fc7c8fc561cf2be","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2019-05-31 17:33:55.000000000","message":"Patch Set 18:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":18},{"id":"030088383de9b318a6b83ddcd9718073864cddc5","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-03 10:17:13.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"4afd5108a5cba941d8f036eb0d9ef4ca6185aa77","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-06-03 10:29:42.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":19},{"id":"c41d25029abc9e25dda7bf7cb728cba99bbbcf73","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-03 10:53:28.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/19/check/dsvm-tempest-kvm/f8cf971 : FAILURE in 35m 13s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/19/check/dsvm-tempest-exe/3c4cdce : FAILURE in 34m 55s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":19},{"id":"339f3474631f234e8527af267481b63db4f5febb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-03 11:09:22.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/19/iscsi/ : SUCCESS in 51m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/19/smb/ : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":19},{"id":"bbd79d8ac5f2080a23558d3e77000e4990525f34","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-06-03 11:09:49.000000000","message":"Patch Set 19:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-06-03_03-17-20/ : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"5809a498b624685dc9820cbe339475bb6f10267c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-03 11:27:49.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/19/check/dsvm-tempest-cisco-zonemanager-job/f85ddc9/ : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":19},{"id":"c7c3d141ac92c0f3e54ce9fa636b71b3da7a9d4d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-03 11:29:49.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/19/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/7e506ae : SUCCESS in 52m 30s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/19/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/aad3770 : SUCCESS in 1h 12m 15s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":19},{"id":"1fd2be4b0cc1840571a4ae0a8ba161df73f619cb","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-06-03 11:41:25.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/19/check/DellEMC_XtremIO_FC/8201be6/DellEMC_XtremIO_FC/486 : SUCCESS in 52m 56s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/19/check/DellEMC_XtremIO_iSCSI/87a9064/DellEMC_XtremIO_iSCSI/454 : SUCCESS in 1h 23m 49s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":19},{"id":"d650d3fab4789885aa8cccc0ff31c4460b373316","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-03 11:45:37.000000000","message":"Patch Set 19:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/19 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"586b073e8fd563b89ef09826b96b552597e7386e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-03 11:46:13.000000000","message":"Patch Set 19:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/19/check/tempest-dsvm-full-xenial/bbed2ba/ : SUCCESS in 1h 21m 20s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/19/check/devstack-plugin-ceph-dsvm-xenial/7199f92/ : SUCCESS in 1h 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"138859b910495742060da936be811739e4c161dd","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-06-03 11:53:10.000000000","message":"Patch Set 19:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/19/check/fujitsu-eternus-dx-iscsi/da5c22e : SUCCESS in 1h 33m 02s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/19/check/fujitsu-eternus-dx-fc/633c32c : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":19},{"id":"4fa63d10e85f2d4b277a6c85d2b7a87db0ed69f6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-06-03 11:53:48.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/85/570885/19/check/dsvm-tempest-zadara-driver/92378ec : FAILURE in 1h 36m 19s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"0a066e25bc66ba2bbcee0fd4c578bb974c935b8e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-03 11:56:39.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/19/check/dsvm-tempest-infinibox-fc/2e9958c : SUCCESS in 1h 38m 52s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/19/check/dsvm-tempest-infinibox-iscsi/e1fb146 : SUCCESS in 1h 18m 38s","accounts_in_message":[],"_revision_number":19},{"id":"444034458e7f8bd2641b64055ff235f6901d1bff","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-06-03 12:26:25.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/19/check/NEC-ISCSI/724e985 : SUCCESS in 1h 52m 13s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/19/check/NEC-FC/05beb1f : SUCCESS in 1h 52m 48s","accounts_in_message":[],"_revision_number":19},{"id":"e74e48782257cc87a2a0ac95617bb37f1c04d1be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 12:57:32.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/19/check/openstack-tox-lower-constraints/5268c10/ : SUCCESS in 13m 39s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/19/check/openstack-tox-pep8/ee8b2a1/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/19/check/openstack-tox-py27/0691b03/ : SUCCESS in 8m 05s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/19/check/openstack-tox-py36/fdb9188/ : SUCCESS in 9m 42s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/19/check/openstack-tox-py37/7618595/ : SUCCESS in 8m 40s\n- openstack-tox-docs http://logs.openstack.org/85/570885/19/check/openstack-tox-docs/12b5360/html/ : SUCCESS in 8m 49s\n- tempest-full http://logs.openstack.org/85/570885/19/check/tempest-full/fcd5fbd/ : FAILURE in 12m 18s\n- neutron-grenade http://logs.openstack.org/85/570885/19/check/neutron-grenade/97dffb2/ : FAILURE in 3m 49s\n- grenade-py3 http://logs.openstack.org/85/570885/19/check/grenade-py3/83b4eaa/ : FAILURE in 4m 05s\n- tempest-full-py3 http://logs.openstack.org/85/570885/19/check/tempest-full-py3/5dc9b68/ : FAILURE in 12m 13s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/19/check/cinder-tempest-dsvm-lvm-lio-barbican/bbc9af7/ : SUCCESS in 1h 33m 53s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/19/check/cinder-tox-bandit-baseline/cc27da7/ : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/19/check/openstack-tox-functional-py36/337dfbc/ : SUCCESS in 4m 32s\n- cinder-rally-task http://logs.openstack.org/85/570885/19/check/cinder-rally-task/1d4935e/ : SUCCESS in 39m 19s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/19/check/openstack-tox-pylint/ccb872f/ : FAILURE in 3m 09s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/19/check/devstack-plugin-ceph-tempest/c2d1652/ : SUCCESS in 1h 41m 58s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/19/check/openstack-tox-functional/9368e0a/ : SUCCESS in 6m 21s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/19/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/ec728fe/ : SUCCESS in 1h 05m 20s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/19/check/legacy-tempest-dsvm-lvm-multibackend/fdb2dfd/ : SUCCESS in 1h 34m 05s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/19/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/517001c/ : FAILURE in 1h 17m 03s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/19/check/tempest-slow-py3/dc94a7d/ : SUCCESS in 2h 13m 53s","accounts_in_message":[],"_revision_number":19},{"id":"f3455a516fcc48c4710f8a58dff2dd61a941ad03","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-03 13:25:21.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/85/570885/19/check/DellEMC_Unity_FC/ba45d23/DellEMC_Unity_FC/356 : SUCCESS in 1h 07m 47s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/19/check/DellEMC_Unity_iSCSI/0f38f5b/DellEMC_Unity_iSCSI/325 : FAILURE in 2h 56m 55s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"33c26aa494a40477ee909f8470bf876a395a0649","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-03 13:45:48.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"f030ce7194bb40422e27c363315d58e7cf207ef7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-03 14:24:20.000000000","message":"Patch Set 19:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/19/check-not-vote/ext-cinder-zuul/5200fd8 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":19},{"id":"29e95aad5c6c83e5c472e5816253ef26936d8a80","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-03 14:32:35.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/19/check/dsvm-tempest-kvm/9c4a7fd : FAILURE in 40m 58s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/19/check/dsvm-tempest-exe/41ac9f8 : FAILURE in 45m 00s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":19},{"id":"dc7dde1423a336586d8f41ea86124c7b3892abd3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-03 14:41:28.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/19/iscsi/ : FAILURE in 55m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/19/smb/ : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":19},{"id":"675eafb871b28ae69db544ec775bd94251ee253e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-03 15:06:29.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/19/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/3faf4d0 : SUCCESS in 55m 27s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/19/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/18ae618 : SUCCESS in 1h 14m 09s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":19},{"id":"b8c012a1fd5e93389ff10427aa9d00643498564a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-03 15:28:17.000000000","message":"Patch Set 19:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/19/check/tempest-dsvm-full-xenial/a01121c/ : SUCCESS in 1h 14m 05s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/19/check/devstack-plugin-ceph-dsvm-xenial/6e39b74/ : SUCCESS in 1h 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"fa8c971a2ca6461cca50fc6e31f70044cc2b9d6f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-03 15:36:05.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/19/check/dsvm-tempest-cisco-zonemanager-job/1cc7e9c/ : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":19},{"id":"590b988745111b9cb8a86eabbc8e43220d411d35","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-06-03 16:09:08.000000000","message":"Patch Set 19:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/85/570885/19/check/kaminario-dsvm-tempest-full-iscsi/ab3bbd9 : SUCCESS in 2h 00m 22s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/85/570885/19/check/kaminario-dsvm-tempest-full-FC/04f8b29 : FAILURE in 1m 25s","accounts_in_message":[],"_revision_number":19},{"id":"addc13eb7b5781726166184512436441183d96fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 16:29:11.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/19/check/openstack-tox-lower-constraints/48cd174/ : SUCCESS in 13m 19s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/19/check/openstack-tox-pep8/346e01b/ : SUCCESS in 7m 10s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/19/check/openstack-tox-py27/c3453ae/ : SUCCESS in 8m 34s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/19/check/openstack-tox-py36/458e223/ : SUCCESS in 8m 07s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/19/check/openstack-tox-py37/7dec5a6/ : SUCCESS in 8m 10s\n- openstack-tox-docs http://logs.openstack.org/85/570885/19/check/openstack-tox-docs/eb1186a/html/ : SUCCESS in 7m 59s\n- tempest-full http://logs.openstack.org/85/570885/19/check/tempest-full/8eed314/ : SUCCESS in 1h 57m 47s\n- neutron-grenade http://logs.openstack.org/85/570885/19/check/neutron-grenade/53e577f/ : SUCCESS in 53m 55s\n- grenade-py3 http://logs.openstack.org/85/570885/19/check/grenade-py3/768a058/ : SUCCESS in 1h 02m 43s\n- tempest-full-py3 http://logs.openstack.org/85/570885/19/check/tempest-full-py3/a956904/ : SUCCESS in 1h 38m 11s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/19/check/cinder-tempest-dsvm-lvm-lio-barbican/e905a2c/ : SUCCESS in 1h 48m 03s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/19/check/cinder-tox-bandit-baseline/499f04d/ : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/19/check/openstack-tox-functional-py36/a9e4774/ : SUCCESS in 5m 18s\n- cinder-rally-task http://logs.openstack.org/85/570885/19/check/cinder-rally-task/5cf62ca/ : SUCCESS in 41m 04s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/19/check/openstack-tox-pylint/08e3b63/ : SUCCESS in 5m 04s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/19/check/devstack-plugin-ceph-tempest/8ce7351/ : SUCCESS in 1h 34m 29s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/19/check/openstack-tox-functional/b8ade2e/ : SUCCESS in 5m 03s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/19/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/2969dde/ : RETRY_LIMIT in 3m 03s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/19/check/legacy-tempest-dsvm-lvm-multibackend/6560512/ : SUCCESS in 1h 35m 17s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/19/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/8b8ddd6/ : FAILURE in 1h 28m 33s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/19/check/tempest-slow-py3/0b66667/ : SUCCESS in 2h 26m 42s","accounts_in_message":[],"_revision_number":19},{"id":"b725f9c09a637768418f65e11fb4461839b5549a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-06-03 17:00:21.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/85/570885/19/check/dsvm-tempest-zadara-driver/ec3c098 : FAILURE in 1h 37m 30s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"cf2505680c97a6056a7fba6e204bdd8db0c1911e","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-06-03 20:04:36.000000000","message":"Patch Set 19:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/85/570885/19/check/kaminario-dsvm-tempest-full-iscsi/ffe42f2 : FAILURE in 1h 48m 56s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/85/570885/19/check/kaminario-dsvm-tempest-full-FC/30b5f64 : SUCCESS in 1h 45m 38s","accounts_in_message":[],"_revision_number":19},{"id":"a090298c74f066c1586d1ccb3a456a60be647e83","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-05 12:55:27.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"135376b181362710dff90aaa2be23805db8a2706","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-05 13:50:15.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/20/iscsi/ : SUCCESS in 54m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/20/smb/ : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":20},{"id":"5225b41995c749a128fc1306bbb61dcb755ac758","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-05 14:11:17.000000000","message":"Patch Set 20:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"6d3a3da3f15f96bd2dccd1791040d8d8bb7bb369","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-05 14:22:33.000000000","message":"Patch Set 20:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/20/check/tempest-dsvm-full-xenial/916f632/ : SUCCESS in 1h 22m 26s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/20/check/devstack-plugin-ceph-dsvm-xenial/1f3f5ee/ : SUCCESS in 1h 26m 44s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"02c814f21377dc4d8302015581c1b0f5c2f06d59","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-06-05 14:30:07.000000000","message":"Patch Set 20:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/20/check/fujitsu-eternus-dx-iscsi/589ac4a : SUCCESS in 1h 31m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/20/check/fujitsu-eternus-dx-fc/c772f3b : SUCCESS in 1h 33m 52s","accounts_in_message":[],"_revision_number":20},{"id":"c20075597f0f580e1969c9087224f56a7ece817c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-05 15:11:07.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/20/check-not-vote/ext-cinder-zuul/8928385 : SUCCESS in 40m 18s","accounts_in_message":[],"_revision_number":20},{"id":"f3ae65d663210647b83e834df0cd7846a18adc5f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-05 15:25:19.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/20/check/dsvm-tempest-infinibox-fc/ffbaf26 : SUCCESS in 1h 06m 09s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/20/check/dsvm-tempest-infinibox-iscsi/b8ece7d : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":20},{"id":"c0ce16845378cef02c920e1ed4ee22bd84d47bc1","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-06-05 15:31:32.000000000","message":"Patch Set 20:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-06-05_07-38-18/ : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"7f005852548bece1d48f3a0b2737e252a92daa03","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-05 15:32:56.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"a574f34948a486bb7d9985ea943fe581750be54d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-05 16:26:32.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/21/iscsi/ : SUCCESS in 53m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/21/smb/ : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":21},{"id":"c09067c4e77dac472dcad6235418e9e081b267aa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-05 16:43:15.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/21/check-not-vote/ext-cinder-zuul/3567996 : FAILURE in 17m 23s","accounts_in_message":[],"_revision_number":21},{"id":"dd3105efb6c78f2a6167ce65944e7b00c9f6abcb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-05 16:53:10.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/21/check/dsvm-tempest-infinibox-fc/1a3abed : SUCCESS in 56m 09s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/21/check/dsvm-tempest-infinibox-iscsi/2fe2e4b : SUCCESS in 1h 19m 42s","accounts_in_message":[],"_revision_number":21},{"id":"9592fc35087de3b428fc5fc20a41e4da8809adcf","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-05 17:12:52.000000000","message":"Patch Set 21:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/21 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"4925a3e2d48a76058bfb94b8566a859d04dfb546","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-06-05 17:16:14.000000000","message":"Patch Set 21:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-06-05_09-24-26/ : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"f1c2cfdc37282a8bda16962b679f965e95cdd682","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-06-05 17:29:11.000000000","message":"Patch Set 21:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/21/check/fujitsu-eternus-dx-iscsi/d371d75 : SUCCESS in 1h 55m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/21/check/fujitsu-eternus-dx-fc/e30b118 : SUCCESS in 1h 51m 50s","accounts_in_message":[],"_revision_number":21},{"id":"267b6f5433f1f702328ef9a1725ee35d2d89d069","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-05 17:34:27.000000000","message":"Patch Set 21:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/21/check/tempest-dsvm-full-xenial/85984ae/ : SUCCESS in 2h 01m 08s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/21/check/devstack-plugin-ceph-dsvm-xenial/4429a2e/ : SUCCESS in 1h 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"cee935b2f6dcf350391305c346399c9eb28dbf1c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-05 17:45:35.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/21/check/dsvm-tempest-kvm/9bea9c1 : FAILURE in 37m 15s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/21/check/dsvm-tempest-exe/e97030d : FAILURE in 45m 11s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"45529dde8eed282940bdb5c00470d571bf8010e0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-05 19:37:20.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/21/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/af36e92 : SUCCESS in 59m 17s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/21/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/7dab04a : SUCCESS in 1h 14m 30s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":21},{"id":"e58b7a7a72dd56587b7265143e860586c39d6d29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 22:13:02.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/21/check/openstack-tox-lower-constraints/4b9dbcf/ : SUCCESS in 10m 39s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/21/check/openstack-tox-pep8/d43559d/ : SUCCESS in 7m 33s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/21/check/openstack-tox-py27/b500af6/ : SUCCESS in 8m 15s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/21/check/openstack-tox-py36/99b0f60/ : SUCCESS in 9m 16s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/21/check/openstack-tox-py37/c5eb6f5/ : SUCCESS in 10m 34s\n- openstack-tox-docs http://logs.openstack.org/85/570885/21/check/openstack-tox-docs/c826108/html/ : SUCCESS in 8m 28s\n- tempest-full http://logs.openstack.org/85/570885/21/check/tempest-full/4f94dfd/ : SUCCESS in 1h 51m 08s\n- neutron-grenade http://logs.openstack.org/85/570885/21/check/neutron-grenade/5f76d00/ : SUCCESS in 1h 07m 47s\n- grenade-py3 http://logs.openstack.org/85/570885/21/check/grenade-py3/2a18d6e/ : SUCCESS in 58m 21s\n- tempest-full-py3 http://logs.openstack.org/85/570885/21/check/tempest-full-py3/93561f3/ : SUCCESS in 1h 32m 41s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/21/check/cinder-tempest-dsvm-lvm-lio-barbican/5687361/ : SUCCESS in 1h 48m 14s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/21/check/cinder-tox-bandit-baseline/21ecc07/ : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/21/check/openstack-tox-functional-py36/684e062/ : SUCCESS in 6m 08s\n- cinder-rally-task http://logs.openstack.org/85/570885/21/check/cinder-rally-task/a218c32/ : SUCCESS in 37m 57s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/21/check/openstack-tox-pylint/6bff108/ : SUCCESS in 4m 42s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/21/check/devstack-plugin-ceph-tempest/7a4fb2b/ : SUCCESS in 1h 32m 49s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/21/check/openstack-tox-functional/9ca33bc/ : SUCCESS in 4m 40s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/21/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/1a7e144/ : SUCCESS in 1h 06m 51s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/21/check/legacy-tempest-dsvm-lvm-multibackend/8df8088/ : SUCCESS in 1h 19m 14s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/21/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/f926c15/ : SUCCESS in 1h 38m 00s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/21/check/tempest-slow-py3/c385c90/ : SUCCESS in 2h 19m 19s","accounts_in_message":[],"_revision_number":21},{"id":"12a1806d65b6d50f560981ca3bf24849030e67a0","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-06-05 22:21:37.000000000","message":"Patch Set 21:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-85-570885-21/nedge_iscsi : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":21},{"id":"e4508a8fca3cc5ceff60dfc17015cc69408d7dd0","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2019-06-05 22:26:40.000000000","message":"Patch Set 21:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-85-570885-21/index.html : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"df82fd3a85dc9c3eb31197059664738d0e321ec4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-06-05 22:50:02.000000000","message":"Patch Set 21:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-85-570885-21/ns5_iscsi : FAILURE in 6m 16s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-85-570885-21/ns5_nfs : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":21},{"id":"02a7271d60300dbbdb67cda64d5ce8b2efb6bbaa","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-06-05 23:13:21.000000000","message":"Patch Set 21:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/2193 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"e9db7b98bb2de17df63cbd865da82e1009ad28f7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-05 23:27:01.000000000","message":"Patch Set 21:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-21 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":21},{"id":"b14945d7679a6f6a0f66d0a141e32dfbdeecd12c","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-06-05 23:40:31.000000000","message":"Patch Set 21:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-85-570885-21-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"e99e111d8e7bf3fbf24ad4cbedd274238f1ff3fb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-06 00:16:19.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/85/570885/21/check/DellEMC_Unity_FC/04dc9e1/DellEMC_Unity_FC/376 : SUCCESS in 1h 08m 21s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/21/check/DellEMC_Unity_iSCSI/1d43274/DellEMC_Unity_iSCSI/347 : SUCCESS in 1h 18m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"01e0f8795d970c39491e5cf384945d277c7af894","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-06 00:28:17.000000000","message":"Patch Set 21:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/21/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f4597f7 : FAILURE in 2h 12m 37s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/21/check/3par-fc-driver-master-client-pip-ssa02-dsvm/5195aea : FAILURE in 2h 10m 18s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/85/570885/21/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/f70f5d2 : SUCCESS in 2h 07m 33s","accounts_in_message":[],"_revision_number":21},{"id":"c254747bd8e69a104b7695ef2cb640682e6ca42a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-06-06 00:35:14.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/85/570885/21/check/DellEMC_SC_FC/1353d65/DellEMC_SC_FC/109 : SUCCESS in 1h 19m 18s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/85/570885/21/check/DellEMC_SC_iSCSI/f16e003/DellEMC_SC_iSCSI/106 : FAILURE in 1h 47m 33s\n\nLeave a comment with \u0027run-dellemc-sc\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":21},{"id":"1b86da9dd86656bd86c54852cabe7e57a5d53db0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-06-06 01:05:58.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/85/570885/21/upstream-solidfire-check/cinder-solidfire-iSCSI/732f132/ : FAILURE in 2h 51m 44s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":21},{"id":"cc3fd6c9a36c28c5f47b8b86755f471462427cdd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-06 02:06:34.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/85/570885/21/upstream-check/cinder-cDOT-iSCSI/01d4483/ : FAILURE in 3h 52m 48s\n- cinder-cDOT-NFS http://13.56.179.158/logs/85/570885/21/upstream-check/cinder-cDOT-NFS/1328a6c/ : FAILURE in 2h 36m 19s\n- cinder-cDOT-FCP http://13.56.179.158/logs/85/570885/21/upstream-check/cinder-cDOT-FCP/011130c/ : FAILURE in 3h 50m 35s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":21},{"id":"c1c8d8108020dec17ae635035f284ec80c51467a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-06 02:33:30.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/21/check/dsvm-tempest-cisco-zonemanager-job/236a10c/ : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":21},{"id":"3fe787ce069d25ff43fb387222c656c1af3c22ed","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-06-06 06:53:41.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/21/check/DellEMC_XtremIO_FC/6fbde81/DellEMC_XtremIO_FC/534 : SUCCESS in 56m 21s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/21/check/DellEMC_XtremIO_iSCSI/b9155e6/DellEMC_XtremIO_iSCSI/496 : SUCCESS in 1h 20m 20s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":21},{"id":"ec0b048d33970543d5cb1888ffc5db14cf431076","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-06 07:11:43.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/85/570885/21/check/DellEMC_VNX_iSCSI/a223210/DellEMC_VNX_iSCSI/274 : SUCCESS in 1h 42m 51s\n- DellEMC_VNX_FC http://publiclogs.emc.com/85/570885/21/check/DellEMC_VNX_FC/6d98f19/DellEMC_VNX_FC/281 : SUCCESS in 1h 03m 55s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":21},{"id":"1736e91c54bd40b2df8279298aab3a5ffe3bdf80","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-06 10:18:16.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"7217ad63df647986ea41503fe96851a27c0a9ba4","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-06-06 10:18:50.000000000","message":"Patch Set 22:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1456 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1509 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1670 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"56d0742b3a45044736cc92b830fecdb75f68d625","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-06 10:18:51.000000000","message":"Patch Set 22:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/22/check/dsvm-tempest-infinibox-fc/0d750c9 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/22/check/dsvm-tempest-infinibox-iscsi/b11a2cc : NOT_REGISTERED","accounts_in_message":[],"_revision_number":22},{"id":"ebde1edfbe67c362c19dafec4b75cc82248f8451","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-06 11:12:27.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/22/iscsi/ : FAILURE in 53m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/22/smb/ : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":22},{"id":"e6e25f296e1c6ea5b1e7cccfa51e6520a5e915b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-06 11:39:23.000000000","message":"Patch Set 22:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/22/check/tempest-dsvm-full-xenial/a63af65/ : FAILURE in 1h 17m 38s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/22/check/devstack-plugin-ceph-dsvm-xenial/7ecc2bc/ : FAILURE in 1h 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"2c394004a20e68794ca23c93e153e4584504f38a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-06 12:49:01.000000000","message":"Patch Set 22:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/22/check-not-vote/ext-cinder-zuul/e9c924a : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":22},{"id":"d861577cbb017bcaf43d20c59eb0495b0c01b9af","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-06-06 12:57:10.000000000","message":"Patch Set 22:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-06-06_05-04-19/ : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"5482b6e94878aab75c7578683fe7ff76030b9078","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-06-06 13:14:54.000000000","message":"Patch Set 22:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/22/check/fujitsu-eternus-dx-iscsi/83115f3 : SUCCESS in 2h 16m 06s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/22/check/fujitsu-eternus-dx-fc/e416e54 : SUCCESS in 2h 13m 05s","accounts_in_message":[],"_revision_number":22},{"id":"67589122d482c8e5fb6e71375db0c9f36e8a1f30","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-06 13:21:33.000000000","message":"Patch Set 22:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"2c60a7d543ec55208fd6e69c045b7b053bd9a044","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-06 13:23:39.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/22/check/dsvm-tempest-kvm/2c7dff4 : FAILURE in 41m 05s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/22/check/dsvm-tempest-exe/ac69be7 : FAILURE in 34m 17s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"5d848adbdd5a25150883a0cb0f595907699eb4e6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-06 14:55:27.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/85/570885/22/check/DellEMC_Unity_FC/482a13d/DellEMC_Unity_FC/385 : SUCCESS in 1h 21m 41s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/22/check/DellEMC_Unity_iSCSI/07f485c/DellEMC_Unity_iSCSI/356 : SUCCESS in 1h 16m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"4ddab28d6f27441847ef5c3e49bfbb4195f1436c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-06 14:59:03.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/570885/22/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/7b2bb95 : SUCCESS in 55m 53s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/570885/22/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/f9fb1ea : SUCCESS in 1h 06m 46s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":22},{"id":"e902c3e320b2fb4d0b8324d7aeb2da6202de972a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:38:09.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/22/check/openstack-tox-lower-constraints/c810e76/ : SUCCESS in 12m 05s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/22/check/openstack-tox-pep8/673ce6d/ : SUCCESS in 7m 26s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/22/check/openstack-tox-py27/9cd21a9/ : SUCCESS in 10m 44s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/22/check/openstack-tox-py36/c1ff26c/ : SUCCESS in 9m 02s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/22/check/openstack-tox-py37/817a6e3/ : SUCCESS in 11m 22s\n- openstack-tox-docs http://logs.openstack.org/85/570885/22/check/openstack-tox-docs/2fcfe65/html/ : SUCCESS in 7m 53s\n- tempest-full http://logs.openstack.org/85/570885/22/check/tempest-full/5a00c3f/ : SUCCESS in 1h 40m 51s\n- neutron-grenade http://logs.openstack.org/85/570885/22/check/neutron-grenade/ba887ba/ : SUCCESS in 58m 24s\n- grenade-py3 http://logs.openstack.org/85/570885/22/check/grenade-py3/2378f8f/ : SUCCESS in 58m 10s\n- tempest-full-py3 http://logs.openstack.org/85/570885/22/check/tempest-full-py3/09413ba/ : SUCCESS in 1h 29m 07s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/22/check/cinder-tempest-dsvm-lvm-lio-barbican/86260a3/ : FAILURE in 1h 43m 46s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/22/check/cinder-tox-bandit-baseline/aa87c26/ : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/22/check/openstack-tox-functional-py36/b89ed91/ : SUCCESS in 4m 48s\n- cinder-rally-task http://logs.openstack.org/85/570885/22/check/cinder-rally-task/5a86093/ : SUCCESS in 37m 32s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/22/check/openstack-tox-pylint/d303b62/ : SUCCESS in 5m 57s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/22/check/devstack-plugin-ceph-tempest/1ef86bd/ : SUCCESS in 1h 31m 01s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/22/check/openstack-tox-functional/3af2008/ : SUCCESS in 5m 42s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/22/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/98aecef/ : SUCCESS in 1h 08m 12s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/22/check/legacy-tempest-dsvm-lvm-multibackend/5c915b3/ : SUCCESS in 1h 26m 38s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/22/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/47233d1/ : SUCCESS in 1h 43m 08s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/22/check/tempest-slow-py3/f445df2/ : SUCCESS in 2h 17m 55s","accounts_in_message":[],"_revision_number":22},{"id":"c3c83f4b89c2cafa434134fad5a5421d915bed60","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-06-06 18:36:37.000000000","message":"Patch Set 22: Code-Review-1\n\nI think this should have a release note.\n\nOtherwise this looks good.","accounts_in_message":[],"_revision_number":22},{"id":"f2318cd0cab22fc446bb1f1ba420bafbe973869d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-06 20:40:56.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/22/check/dsvm-tempest-cisco-zonemanager-job/e689235/ : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":22},{"id":"5fb387b54c67999c4ca0e4b6b7093f3b96c511d2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 00:07:30.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/22/check/DellEMC_XtremIO_FC/a6c7608/DellEMC_XtremIO_FC/543 : SUCCESS in 52m 56s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/22/check/DellEMC_XtremIO_iSCSI/864067c/DellEMC_XtremIO_iSCSI/505 : SUCCESS in 1h 39m 11s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":22},{"id":"baea097a4d7e558a6392f507cbc95b6628078e78","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 01:32:00.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/22/check/NEC-ISCSI/e7ab20e : SUCCESS in 1h 58m 43s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/22/check/NEC-FC/aab1bde : SUCCESS in 1h 54m 36s","accounts_in_message":[],"_revision_number":22},{"id":"ce7239fbfecd6320303295b2126108e0efa03858","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-07 09:18:00.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"24255f6064d40ad7cc708ae8eea90eb17d4078ff","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-07 09:18:34.000000000","message":"Patch Set 23:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/23/check/dsvm-tempest-infinibox-fc/a855f46 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/23/check/dsvm-tempest-infinibox-iscsi/82c878f : NOT_REGISTERED","accounts_in_message":[],"_revision_number":23},{"id":"4f34ce95c74b1f324045c40bf9ba5aab95d12459","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-06-07 09:25:55.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1515 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1462 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1676 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"79f7f442ab638fef3f7ab8d34115b294085abc03","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 09:51:48.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/23/check/dsvm-tempest-kvm/81c3036 : FAILURE in 32m 52s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/23/check/dsvm-tempest-exe/065aac1 : FAILURE in 1m 56s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"7592d203ce4c27b1183b728ca8170d3b03915a0c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-07 09:58:38.000000000","message":"Patch Set 23:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/23/check-not-vote/ext-cinder-zuul/bd00c97 : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":23},{"id":"ad2270bcbde41652a185e46cfa2bfa7926de5875","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-07 10:08:56.000000000","message":"Patch Set 23:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/23/iscsi/ : FAILURE in 50m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/23/smb/ : SUCCESS in 48m 15s","accounts_in_message":[],"_revision_number":23},{"id":"d87370fd37c278e416775712b13649c2008e8bd4","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-06-07 10:11:33.000000000","message":"Patch Set 23:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-06-07_02-18-07/ : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"952bdd22870baaa5d224bd83c9194bc3051908a2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-07 10:31:47.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/23/check/dsvm-tempest-cisco-zonemanager-job/1f0da4d/ : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":23},{"id":"f7d1575dd459a492a6ecd19baf000b0d7c3e43b5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-07 10:44:07.000000000","message":"Patch Set 23:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/23/check/tempest-dsvm-full-xenial/869f871/ : FAILURE in 1h 24m 18s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/23/check/devstack-plugin-ceph-dsvm-xenial/3042c3d/ : FAILURE in 1h 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"199e3ce08ae5f8b271fd5fb7d5003d1f226ed7f6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-07 10:51:36.000000000","message":"Patch Set 23:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/23 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"b1b41957b7df5af3b934d7bce5d093c6d43d28a7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-06-07 10:55:42.000000000","message":"Patch Set 23:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/23/check/fujitsu-eternus-dx-iscsi/6595f93 : SUCCESS in 1h 36m 59s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/23/check/fujitsu-eternus-dx-fc/9385a4f : SUCCESS in 1h 34m 38s","accounts_in_message":[],"_revision_number":23},{"id":"2408ffb76f20259e395285b16afef0ec656bedbe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-07 10:57:37.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/85/570885/23/check/DellEMC_Unity_FC/96f0bbe/DellEMC_Unity_FC/399 : SUCCESS in 1h 16m 47s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/23/check/DellEMC_Unity_iSCSI/1136fb9/DellEMC_Unity_iSCSI/370 : SUCCESS in 1h 39m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"7aeb3a46746b4aad05dde2cd06573a8db09ba16b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 11:21:37.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/23/check/NEC-ISCSI/32cc76e : SUCCESS in 1h 27m 39s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/23/check/NEC-FC/fe38a74 : SUCCESS in 1h 30m 01s","accounts_in_message":[],"_revision_number":23},{"id":"2d9d2be7f85e8cb2eae203a00540e6d2c615f156","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 11:49:05.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/23/check/DellEMC_XtremIO_FC/d27097e/DellEMC_XtremIO_FC/550 : SUCCESS in 50m 36s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/23/check/DellEMC_XtremIO_iSCSI/e1f7ffa/DellEMC_XtremIO_iSCSI/512 : SUCCESS in 1h 27m 34s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":23},{"id":"60c5949952024e4ffdf5cc4e7ef96ae96ed95fe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 12:56:37.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/23/check/openstack-tox-lower-constraints/1609abe/ : SUCCESS in 10m 09s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/23/check/openstack-tox-pep8/6696817/ : SUCCESS in 7m 59s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/23/check/openstack-tox-py27/4a0bd9f/ : SUCCESS in 10m 16s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/23/check/openstack-tox-py36/224af3f/ : SUCCESS in 9m 39s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/23/check/openstack-tox-py37/6e81c20/ : SUCCESS in 8m 42s\n- openstack-tox-docs http://logs.openstack.org/85/570885/23/check/openstack-tox-docs/edf167b/html/ : SUCCESS in 7m 17s\n- tempest-full http://logs.openstack.org/85/570885/23/check/tempest-full/69b0a78/ : SUCCESS in 1h 46m 33s\n- neutron-grenade http://logs.openstack.org/85/570885/23/check/neutron-grenade/684df71/ : SUCCESS in 1h 02m 02s\n- grenade-py3 http://logs.openstack.org/85/570885/23/check/grenade-py3/63c7082/ : SUCCESS in 59m 13s\n- tempest-full-py3 http://logs.openstack.org/85/570885/23/check/tempest-full-py3/4faf617/ : SUCCESS in 1h 45m 16s\n- build-openstack-releasenotes http://logs.openstack.org/85/570885/23/check/build-openstack-releasenotes/3fc9991/html/ : SUCCESS in 16m 46s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/23/check/cinder-tempest-dsvm-lvm-lio-barbican/8549263/ : SUCCESS in 1h 34m 22s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/23/check/cinder-tox-bandit-baseline/5b728a8/ : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/23/check/openstack-tox-functional-py36/0660ff1/ : SUCCESS in 5m 04s\n- cinder-rally-task http://logs.openstack.org/85/570885/23/check/cinder-rally-task/e107fde/ : SUCCESS in 40m 03s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/23/check/openstack-tox-pylint/d4fbb56/ : SUCCESS in 5m 05s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/23/check/devstack-plugin-ceph-tempest/cde2d70/ : SUCCESS in 1h 27m 23s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/23/check/openstack-tox-functional/cdce8a6/ : SUCCESS in 4m 40s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/23/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/626912e/ : SUCCESS in 1h 04m 34s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/23/check/legacy-tempest-dsvm-lvm-multibackend/05f3d93/ : SUCCESS in 1h 21m 27s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/23/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/60ba664/ : SUCCESS in 2h 03m 38s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/23/check/tempest-slow-py3/0522d2f/ : SUCCESS in 2h 22m 53s","accounts_in_message":[],"_revision_number":23},{"id":"d7f351ccbfbd1f71aa9bbe13b43073a7828fab07","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-06-07 12:59:04.000000000","message":"Patch Set 23:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-85-570885-23/nedge_iscsi : FAILURE in 0m 9s","accounts_in_message":[],"_revision_number":23},{"id":"227c147124bb646b98591b98b5442af48d4641c7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-06-07 12:59:11.000000000","message":"Patch Set 23:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-85-570885-23/ns5_iscsi : FAILURE in 0m 8s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-85-570885-23/ns5_nfs : FAILURE in 0m 8s","accounts_in_message":[],"_revision_number":23},{"id":"7685e87da5b13927ee28716d533960e15b2a7863","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 13:14:58.000000000","message":"Patch Set 23:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-85-570885-23/index.html : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"fa8a9f9d259813020f4223117a3b8a60309dc1a4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-07 13:48:57.000000000","message":"Patch Set 23:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/23/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/15ab6b0 : FAILURE in 49m 41s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/23/check/3par-fc-driver-master-client-pip-ssa02-dsvm/14c3e41 : FAILURE in 46m 25s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/85/570885/23/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/5e8e754 : FAILURE in 45m 12s","accounts_in_message":[],"_revision_number":23},{"id":"98982eb2cffa46bcdac3442dcea4ad7ea328babb","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-06-07 13:51:53.000000000","message":"Patch Set 23: Code-Review+2\n\nThanks for adding the release note.","accounts_in_message":[],"_revision_number":23},{"id":"5c6080d869cc033c49a3597efb0191fa533e169b","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-06-07 13:53:27.000000000","message":"Patch Set 23:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/2209 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"818a91dcb415e1859dec9768499a86f03aab3308","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 14:12:53.000000000","message":"Patch Set 23:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-85-570885-23-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"60beaedb5ef40eb08bf7752a75f03246eee22845","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-07 14:17:55.000000000","message":"Patch Set 23:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-23 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":23},{"id":"00b70478c4f24341addf40928f2a22150321c157","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-06-07 14:21:33.000000000","message":"Patch Set 23: Code-Review-2\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"01a31b03227e2f94a24295e1a6e2f1b0b5eec86a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-07 14:22:21.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/85/570885/23/check/DellEMC_Unity_FC/07dc10e/DellEMC_Unity_FC/400 : SUCCESS in 1h 16m 25s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/23/check/DellEMC_Unity_iSCSI/3e6a14d/DellEMC_Unity_iSCSI/371 : SUCCESS in 1h 25m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"09d8365b9a15bc3d26f121089509267724c573a5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 14:24:06.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/85/570885/23/check/DellEMC_VxFlexOS/7cc6418/DellEMC_VxFlexOS/283 : SUCCESS in 1h 27m 08s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":23},{"id":"fcffbfb95c001c3ddc00b33784abff57090b4fea","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-06-07 14:33:11.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/85/570885/23/check/DellEMC_SC_FC/5589590/DellEMC_SC_FC/121 : SUCCESS in 1h 04m 11s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/85/570885/23/check/DellEMC_SC_iSCSI/882c69e/DellEMC_SC_iSCSI/118 : SUCCESS in 1h 36m 18s\n\nLeave a comment with \u0027run-dellemc-sc\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":23},{"id":"7afd1caeb3f22113e230b4ee6df4a8c4c92891a4","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-06-07 14:42:49.000000000","message":"Patch Set 23:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/85/570885/23/Lenovo-FC : SUCCESS in 32:30.35\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/85/570885/23/Lenovo-iSCSI : SUCCESS in 33:33.98","accounts_in_message":[],"_revision_number":23},{"id":"63b913d1ed5c240289d38ab9a2c9bc7a61d502e6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-06-07 14:48:17.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/85/570885/23/upstream-solidfire-check/cinder-solidfire-iSCSI/133691e/ : SUCCESS in 1h 50m 23s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":23},{"id":"7628e050090d08c59f525490ea249f6028c433b2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-07 14:51:03.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/85/570885/23/upstream-check/cinder-cDOT-iSCSI/f57d0cc/ : SUCCESS in 1h 53m 41s\n- cinder-cDOT-NFS http://13.56.179.158/logs/85/570885/23/upstream-check/cinder-cDOT-NFS/6aafcac/ : SUCCESS in 1h 30m 23s\n- cinder-cDOT-FCP http://13.56.179.158/logs/85/570885/23/upstream-check/cinder-cDOT-FCP/d5cba9b/ : FAILURE in 1h 27m 01s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"60e975e025a3cf65845241b52bd41d115861d1d1","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-06-07 14:51:19.000000000","message":"Patch Set 23:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/85/570885/23/HPMSA-FC : SUCCESS in 37:05.97\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/85/570885/23/HPMSA-iSCSI : SUCCESS in 37:39.65","accounts_in_message":[],"_revision_number":23},{"id":"1084062e540434d1faf72e8a2f5227f7f315f203","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 14:56:29.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/85/570885/23/check/DellEMC_VNX_iSCSI/463d2e6/DellEMC_VNX_iSCSI/286 : FAILURE in 1h 59m 35s\n- DellEMC_VNX_FC http://publiclogs.emc.com/85/570885/23/check/DellEMC_VNX_FC/2e02c93/DellEMC_VNX_FC/293 : SUCCESS in 1h 07m 39s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":23},{"id":"3abd979fd0438edef2f72d76f6c579ef2be8db69","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 15:11:40.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/23/check/DellEMC_XtremIO_FC/260ab8c/DellEMC_XtremIO_FC/551 : SUCCESS in 54m 12s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/23/check/DellEMC_XtremIO_iSCSI/2f58b3b/DellEMC_XtremIO_iSCSI/513 : SUCCESS in 1h 46m 04s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":23},{"id":"dc712a1cc48bdfa364415e48a7e5a221f7beac66","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-11 07:51:55.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"fd3cc59df4bdb9f1ec3ac4a677b41a8185c10606","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-11 07:54:17.000000000","message":"Patch Set 23:\n\n(4 comments)\n\nMoved to obj_context in update.","accounts_in_message":[],"_revision_number":23},{"id":"8de328c6e071c85f7d76e1751bac149e6df37bbb","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-06-11 07:59:28.000000000","message":"Patch Set 24:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1481 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1534 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1693 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"b8c8e04a7260a38c03873d05b13fefaea2357fba","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 08:28:20.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/24/check/dsvm-tempest-kvm/4459d37 : FAILURE in 33m 28s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/24/check/dsvm-tempest-exe/d683603 : FAILURE in 34m 53s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"4bff2030a6721e0a137b0df93a3940090e24a99e","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-06-11 08:48:03.000000000","message":"Patch Set 24:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-06-11_00-52-02/ : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"fbf2bed73e392849f80dd891824a7c67340af843","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-11 09:07:33.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/24/check/dsvm-tempest-cisco-zonemanager-job/6aa92dd/ : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":24},{"id":"83162af8acdc898909688cd9a382c1a50c843877","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-06-11 09:18:55.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/24/check-cinder/Cinder-ISER-ISCSI/66208ca : SUCCESS in 1h 03m 59s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/24/check-cinder/Cinder-ISER-LIO/1f64bb7 : SUCCESS in 1h 12m 49s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":24},{"id":"d28b10a2763348bd31487909d4aa86584a2d5480","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-11 09:23:08.000000000","message":"Patch Set 24:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/24/check/dsvm-tempest-infinibox-fc/499b981 : SUCCESS in 1h 30m 33s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/24/check/dsvm-tempest-infinibox-iscsi/eedbb90 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":24},{"id":"52f765286eeaccf9d5581181f9be6b661d5f7268","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-06-11 09:27:31.000000000","message":"Patch Set 24:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/24/check/fujitsu-eternus-dx-iscsi/aa879af : SUCCESS in 1h 32m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/24/check/fujitsu-eternus-dx-fc/cdec4cc : SUCCESS in 1h 34m 50s","accounts_in_message":[],"_revision_number":24},{"id":"6284e143266ceb14003cffe231f069c96b7aea0c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-11 09:32:54.000000000","message":"Patch Set 24:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"39f3f2d254cdec709653fb3fc0aa0955b76cbbaf","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 09:55:00.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/24/check/NEC-ISCSI/5613965 : SUCCESS in 1h 58m 31s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/24/check/NEC-FC/0021e12 : SUCCESS in 2h 02m 20s","accounts_in_message":[],"_revision_number":24},{"id":"2a0af15bc8c7bbed4f75263c7ba8db339edc601b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 10:29:56.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/24/check/DellEMC_XtremIO_FC/2fdea32/DellEMC_XtremIO_FC/574 : SUCCESS in 53m 07s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/24/check/DellEMC_XtremIO_iSCSI/8e69374/DellEMC_XtremIO_iSCSI/536 : SUCCESS in 2h 37m 30s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":24},{"id":"a12e873d54a18c00ff39582b4fac00e1e7ceeca2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:15:20.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/24/check/openstack-tox-lower-constraints/03600cf/ : SUCCESS in 10m 47s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/24/check/openstack-tox-pep8/3735492/ : SUCCESS in 7m 48s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/24/check/openstack-tox-py27/eb088d4/ : SUCCESS in 9m 33s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/24/check/openstack-tox-py36/95d1bc8/ : SUCCESS in 10m 42s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/24/check/openstack-tox-py37/73ebcc8/ : FAILURE in 8m 50s\n- openstack-tox-docs http://logs.openstack.org/85/570885/24/check/openstack-tox-docs/fe60e8f/html/ : SUCCESS in 8m 39s\n- tempest-full http://logs.openstack.org/85/570885/24/check/tempest-full/79668c1/ : SUCCESS in 2h 03m 01s\n- neutron-grenade http://logs.openstack.org/85/570885/24/check/neutron-grenade/634c088/ : SUCCESS in 59m 56s\n- grenade-py3 http://logs.openstack.org/85/570885/24/check/grenade-py3/2403180/ : SUCCESS in 47m 26s\n- tempest-full-py3 http://logs.openstack.org/85/570885/24/check/tempest-full-py3/44cd99a/ : SUCCESS in 1h 26m 55s\n- build-openstack-releasenotes http://logs.openstack.org/85/570885/24/check/build-openstack-releasenotes/10932cc/html/ : SUCCESS in 16m 56s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/24/check/cinder-tempest-dsvm-lvm-lio-barbican/c3f27cd/ : SUCCESS in 1h 42m 36s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/24/check/cinder-tox-bandit-baseline/c744642/ : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/24/check/openstack-tox-functional-py36/f2b4445/ : SUCCESS in 5m 35s\n- cinder-rally-task http://logs.openstack.org/85/570885/24/check/cinder-rally-task/d3d9d72/ : SUCCESS in 40m 49s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/24/check/openstack-tox-pylint/ea9f40f/ : SUCCESS in 4m 24s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/24/check/devstack-plugin-ceph-tempest/cc4b69f/ : SUCCESS in 1h 28m 13s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/24/check/openstack-tox-functional/919598c/ : SUCCESS in 6m 23s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/24/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/fca45d3/ : SUCCESS in 1h 08m 09s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/24/check/legacy-tempest-dsvm-lvm-multibackend/5bc4922/ : SUCCESS in 1h 27m 51s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/24/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/449aac6/ : SUCCESS in 1h 34m 25s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/24/check/tempest-slow-py3/ef91561/ : SUCCESS in 2h 20m 20s","accounts_in_message":[],"_revision_number":24},{"id":"d83400e222a89eb80d757a0c897d23c46ef83b5b","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-11 12:59:37.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"b93e670f7caad3f0089b8ad553d2ac3bd7bb8cd9","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-11 14:14:30.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"aeaa8faaadfd3256e3f521628570493243356b27","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-06-11 14:21:53.000000000","message":"Patch Set 25:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1490 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1543 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1702 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"39560f17c3c01d8b4059c0155fb5d0e0e2ec2d38","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 14:56:38.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/25/check/dsvm-tempest-kvm/517a807 : FAILURE in 38m 25s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/25/check/dsvm-tempest-exe/0d56452 : FAILURE in 39m 38s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":25},{"id":"129ab2d71acd4787afd1dd31d5af016692a9a1d7","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-06-11 15:34:32.000000000","message":"Patch Set 25:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-06-11_07-31-16/ : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"5ab0a127b1d225040a98dc4bff779e301009374b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-06-11 15:45:13.000000000","message":"Patch Set 25:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/25/check/fujitsu-eternus-dx-iscsi/270864a : SUCCESS in 1h 30m 00s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/25/check/fujitsu-eternus-dx-fc/d4e84b0 : SUCCESS in 1h 28m 32s","accounts_in_message":[],"_revision_number":25},{"id":"fa0810d98d62f9db8738ba05f006326cc1e603ed","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-11 15:52:17.000000000","message":"Patch Set 25:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/25 : SUCCESS","accounts_in_message":[],"_revision_number":25},{"id":"4938948d9dd105b477e97bfd7842fb62158f5708","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-11 16:36:24.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/25/check/dsvm-tempest-infinibox-fc/ad52f7e : SUCCESS in 1h 01m 09s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/25/check/dsvm-tempest-infinibox-iscsi/9c2d423 : SUCCESS in 1h 30m 35s","accounts_in_message":[],"_revision_number":25},{"id":"a810b7b803e2abe82203571630e6d5afdc91dba9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-11 17:58:19.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/25/check/dsvm-tempest-cisco-zonemanager-job/f53d5b4/ : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":25},{"id":"9a6130f05fd947af2f734b3bd29b990ddbb3240d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-06-11 20:10:16.000000000","message":"Patch Set 25:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/25/check-cinder/Cinder-ISER-ISCSI/d2a3a06 : FAILURE in 1h 09m 23s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/25/check-cinder/Cinder-ISER-LIO/50f3755 : FAILURE in 17m 16s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":25},{"id":"eb6a9d6a2f7243bb00ce098128e1583dac10ced4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-11 20:54:47.000000000","message":"Patch Set 25:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/25/check-not-vote/ext-cinder-zuul/539c2cf : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":25},{"id":"c7f0300bcc766d3824348305d547ac410a4e62c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:13:17.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/25/check/openstack-tox-lower-constraints/af69d70/ : SUCCESS in 12m 50s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/25/check/openstack-tox-pep8/bdc0c32/ : SUCCESS in 11m 09s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/25/check/openstack-tox-py27/d04f796/ : SUCCESS in 8m 53s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/25/check/openstack-tox-py36/5c450ae/ : SUCCESS in 8m 48s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/25/check/openstack-tox-py37/559cc40/ : SUCCESS in 8m 39s\n- openstack-tox-docs http://logs.openstack.org/85/570885/25/check/openstack-tox-docs/3b383e3/html/ : SUCCESS in 8m 05s\n- tempest-full http://logs.openstack.org/85/570885/25/check/tempest-full/bb856bb/ : TIMED_OUT in 2h 08m 35s\n- neutron-grenade http://logs.openstack.org/85/570885/25/check/neutron-grenade/81c0ea1/ : FAILURE in 15m 51s\n- grenade-py3 http://logs.openstack.org/85/570885/25/check/grenade-py3/8b470c4/ : SUCCESS in 56m 49s\n- tempest-full-py3 http://logs.openstack.org/85/570885/25/check/tempest-full-py3/8bf9c15/ : SUCCESS in 2h 01m 15s\n- build-openstack-releasenotes http://logs.openstack.org/85/570885/25/check/build-openstack-releasenotes/ed2dd85/html/ : SUCCESS in 14m 15s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/25/check/cinder-tempest-dsvm-lvm-lio-barbican/2066a72/ : SUCCESS in 1h 52m 09s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/25/check/cinder-tox-bandit-baseline/88fd72d/ : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/25/check/openstack-tox-functional-py36/b72a283/ : SUCCESS in 5m 53s\n- cinder-rally-task http://logs.openstack.org/85/570885/25/check/cinder-rally-task/fd80773/ : SUCCESS in 41m 29s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/25/check/openstack-tox-pylint/75caf0b/ : SUCCESS in 5m 14s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/25/check/devstack-plugin-ceph-tempest/29b521a/ : SUCCESS in 1h 28m 22s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/25/check/openstack-tox-functional/141c386/ : SUCCESS in 5m 35s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/25/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/cc91d91/ : SUCCESS in 1h 05m 16s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/25/check/legacy-tempest-dsvm-lvm-multibackend/29d5208/ : SUCCESS in 1h 23m 50s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/25/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/96e8d25/ : SUCCESS in 1h 41m 25s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/25/check/tempest-slow-py3/46978fb/ : SUCCESS in 2h 25m 25s","accounts_in_message":[],"_revision_number":25},{"id":"4f22ed30f2104e756f72aad716af557cf271b958","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 23:14:08.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/25/check/NEC-ISCSI/e01cc5f : SUCCESS in 1h 36m 07s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/25/check/NEC-FC/c701521 : SUCCESS in 1h 45m 25s","accounts_in_message":[],"_revision_number":25},{"id":"e6e3490f17a1ff5cdc51f3edc5f41942fd925cd0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-12 00:33:49.000000000","message":"Patch Set 25:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/25/check/tempest-dsvm-full-xenial/9f54472/ : SUCCESS in 1h 21m 24s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/25/check/devstack-plugin-ceph-dsvm-xenial/f171466/ : SUCCESS in 1h 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"8847104649438782dec4c33728fc5b33620fdae3","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-12 07:20:34.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"19a34ffda3f8aa3054f6b28cc5abc3d8fbc3c3d5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 08:02:03.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/25/check/dsvm-tempest-kvm/ba647a3 : FAILURE in 33m 12s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/25/check/dsvm-tempest-exe/be1d5ed : FAILURE in 40m 19s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":25},{"id":"ab116b4ee4e1867ed07c4168a3c5c0e244c6723a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-12 08:45:39.000000000","message":"Patch Set 25:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/25/check/tempest-dsvm-full-xenial/664748a/ : SUCCESS in 1h 19m 47s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/25/check/devstack-plugin-ceph-dsvm-xenial/009764f/ : SUCCESS in 1h 24m 01s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"29a5339c30a0014116536a73df4592ac06d4542e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-12 09:18:00.000000000","message":"Patch Set 25:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/25/check-not-vote/ext-cinder-zuul/60b7a5b : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":25},{"id":"8f7ce58d6ebc05a843860ebbbc4db63240137415","author":{"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},"date":"2019-06-12 11:13:49.000000000","message":"Patch Set 25:\n\nrun INSPUR CI","accounts_in_message":[],"_revision_number":25},{"id":"a660ff3e206c082e835a74b6c8a31d979f274a43","author":{"_account_id":25648,"name":"wang yong","email":"yongwc@foxmail.com","username":"wangyong"},"date":"2019-06-12 11:17:36.000000000","message":"Patch Set 25:\n\nrun-INSPUR CI","accounts_in_message":[],"_revision_number":25},{"id":"d2d9be22c326493bcd5a49f593b7476b1dd9c2ee","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-12 12:03:05.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"692e87c421a3b28f298c77a3c33109afac9d1569","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-12 12:51:49.000000000","message":"Patch Set 25:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/25/check-not-vote/ext-cinder-zuul/cdcc219 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":25},{"id":"7206ee6edd2b5b2fbf4761f38a7fe5b301a305d8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-12 12:57:30.000000000","message":"Patch Set 25:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/25/iscsi/ : SUCCESS in 53m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/25/smb/ : FAILURE in 28m 05s","accounts_in_message":[],"_revision_number":25},{"id":"1ba075bbacc795abb95dade2682e944321ecbb53","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 13:01:58.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/25/check/dsvm-tempest-kvm/55ee2e0 : FAILURE in 58m 11s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/25/check/dsvm-tempest-exe/92d4fbb : FAILURE in 54m 59s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":25},{"id":"55f9cf746e9f6d11c8555c93a26f1c62d3173151","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-06-12 13:34:15.000000000","message":"Patch Set 25:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/85/570885/25/check/Cinder-AS13000/1524fc1 : ABORTED in 2h 05m 16s\n- cinder-test-fc http://218.29.111.78:9000/85/570885/25/check/cinder-test-fc/22ab983 : FAILURE in 9m 17s\n- cinder-test-iscsi http://218.29.111.78:9000/85/570885/25/check/cinder-test-iscsi/297700b : FAILURE in 8m 55s","accounts_in_message":[],"_revision_number":25},{"id":"ac0cfa1c53f10ab43eed747fb282254e29691ec0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-12 14:14:30.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/25/check/dsvm-tempest-cisco-zonemanager-job/f5065f3/ : FAILURE in 37m 50s","accounts_in_message":[],"_revision_number":25},{"id":"0fff432b1918ffaaf3a9a9c3e0d1b01416a2dd20","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-12 14:21:16.000000000","message":"Patch Set 25:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/25/check/tempest-dsvm-full-xenial/79563c9/ : SUCCESS in 1h 58m 08s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/85/570885/25/check/devstack-plugin-ceph-dsvm-xenial/aa2e79d/ : SUCCESS in 1h 22m 31s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"ea0328e89c7651a66d560de88ba2cb977aa7c9a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 15:47:02.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/25/check/openstack-tox-lower-constraints/5eb8ac9/ : SUCCESS in 10m 12s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/25/check/openstack-tox-pep8/1f11713/ : SUCCESS in 7m 51s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/25/check/openstack-tox-py27/c26a720/ : SUCCESS in 8m 11s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/25/check/openstack-tox-py36/f91f57c/ : SUCCESS in 8m 35s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/25/check/openstack-tox-py37/5eda916/ : SUCCESS in 10m 57s\n- openstack-tox-docs http://logs.openstack.org/85/570885/25/check/openstack-tox-docs/2d78871/html/ : SUCCESS in 7m 47s\n- tempest-full http://logs.openstack.org/85/570885/25/check/tempest-full/68a10a2/ : SUCCESS in 1h 44m 37s\n- neutron-grenade http://logs.openstack.org/85/570885/25/check/neutron-grenade/507bcdc/ : SUCCESS in 1h 00m 24s\n- grenade-py3 http://logs.openstack.org/85/570885/25/check/grenade-py3/a3da9fc/ : SUCCESS in 1h 01m 54s\n- tempest-full-py3 http://logs.openstack.org/85/570885/25/check/tempest-full-py3/db6db25/ : SUCCESS in 1h 43m 50s\n- build-openstack-releasenotes http://logs.openstack.org/85/570885/25/check/build-openstack-releasenotes/31b45d2/html/ : SUCCESS in 14m 55s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/25/check/cinder-tempest-dsvm-lvm-lio-barbican/bcd78a3/ : SUCCESS in 1h 49m 38s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/25/check/cinder-tox-bandit-baseline/e256c10/ : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/25/check/openstack-tox-functional-py36/94dc552/ : SUCCESS in 5m 13s\n- cinder-rally-task http://logs.openstack.org/85/570885/25/check/cinder-rally-task/01bd82a/ : SUCCESS in 48m 19s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/25/check/openstack-tox-pylint/920b568/ : SUCCESS in 4m 42s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/85/570885/25/check/devstack-plugin-ceph-tempest/dd370bc/ : SUCCESS in 1h 27m 18s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/25/check/openstack-tox-functional/4d0e430/ : SUCCESS in 4m 48s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/25/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/e2a6343/ : SUCCESS in 1h 09m 41s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/25/check/legacy-tempest-dsvm-lvm-multibackend/1567fef/ : SUCCESS in 1h 36m 35s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/25/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/a664665/ : SUCCESS in 1h 53m 19s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/25/check/tempest-slow-py3/9cef4c3/ : SUCCESS in 2h 28m 02s","accounts_in_message":[],"_revision_number":25},{"id":"8ba2e920b59b0fa7c0590c08fd1db2d488f1b3c9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-06-12 15:49:29.000000000","message":"Patch Set 25:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-85-570885-25/ns5_iscsi : FAILURE in 0m 8s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-85-570885-25/ns5_nfs : FAILURE in 0m 8s","accounts_in_message":[],"_revision_number":25},{"id":"0a98edfce72c129036486629d64c5e9a9837321f","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-06-12 15:49:55.000000000","message":"Patch Set 25:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-85-570885-25/nedge_iscsi : FAILURE in 0m 8s","accounts_in_message":[],"_revision_number":25},{"id":"a6708c698913dcd0817480593615005b5940d96f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-12 16:32:46.000000000","message":"Patch Set 25:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/25/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ca41ffe : FAILURE in 43m 05s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/25/check/3par-fc-driver-master-client-pip-ssa02-dsvm/8358b3d : FAILURE in 39m 45s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/85/570885/25/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/8042e9b : FAILURE in 29m 19s","accounts_in_message":[],"_revision_number":25},{"id":"d4ffc654092fafe17c7421c71837347094edab7d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-12 16:49:30.000000000","message":"Patch Set 25:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-25 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":25},{"id":"0d15781c3b212b530dfbcdb5399b54c0330b4350","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 17:05:40.000000000","message":"Patch Set 25:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-85-570885-25-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":25},{"id":"7dd61d3c767ee88cf6421043b1e7e509297b24cf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 17:12:24.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/85/570885/25/check/DellEMC_VxFlexOS/24bf581/DellEMC_VxFlexOS/302 : SUCCESS in 1h 25m 03s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":25},{"id":"6996eaa9574222671b972e49ec604276ef382988","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-06-12 17:24:56.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/85/570885/25/check/DellEMC_SC_FC/8a98fb6/DellEMC_SC_FC/153 : SUCCESS in 1h 09m 17s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/85/570885/25/check/DellEMC_SC_iSCSI/546488f/DellEMC_SC_iSCSI/147 : SUCCESS in 1h 37m 30s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":25},{"id":"0c4faed208b0893dac3ec00bfdd700b8df2fe663","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-06-12 17:38:59.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/85/570885/25/check/kaminario-dsvm-tempest-full-iscsi/af030fe : SUCCESS in 1h 48m 25s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/85/570885/25/check/kaminario-dsvm-tempest-full-FC/a24e3c1 : SUCCESS in 1h 51m 28s","accounts_in_message":[],"_revision_number":25},{"id":"b2b247e8d777a38f0c9ffa0c83c96f3e4656d54d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-06-12 18:04:08.000000000","message":"Patch Set 25:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/85/570885/25/Lenovo-FC : SUCCESS in 33:05.65\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/85/570885/25/Lenovo-iSCSI : SUCCESS in 34:38.85","accounts_in_message":[],"_revision_number":25},{"id":"4a73ab38b60182f30d460f703a9d33ac878984ab","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-06-12 18:07:41.000000000","message":"Patch Set 25:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/85/570885/25/HPMSA-FC : SUCCESS in 37:29.75\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/85/570885/25/HPMSA-iSCSI : SUCCESS in 38:59.89","accounts_in_message":[],"_revision_number":25},{"id":"e878c8dbe67a0055920879f070bd273945837e89","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 18:41:41.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/85/570885/25/check/DellEMC_VNX_iSCSI/e591f84/DellEMC_VNX_iSCSI/309 : FAILURE in 2h 54m 11s\n- DellEMC_VNX_FC http://publiclogs.emc.com/85/570885/25/check/DellEMC_VNX_FC/3848fee/DellEMC_VNX_FC/316 : SUCCESS in 1h 01m 48s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":25},{"id":"f68f1bc7bf8092bd80cdfe01dcd8f9a1e2b90605","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-12 19:21:38.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/85/570885/25/check/DellEMC_Unity_FC/cc98839/DellEMC_Unity_FC/420 : SUCCESS in 1h 10m 15s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/25/check/DellEMC_Unity_iSCSI/54cc3a8/DellEMC_Unity_iSCSI/391 : SUCCESS in 1h 25m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"aeae5ad10103890704aedcff768f784bdd9b1361","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-12 19:49:27.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/85/570885/25/upstream-check/cinder-cDOT-iSCSI/ab9cdd6/ : FAILURE in 58s\n- cinder-cDOT-NFS http://13.56.179.158/logs/85/570885/25/upstream-check/cinder-cDOT-NFS/93699c2/ : FAILURE in 4h 01m 02s\n- cinder-cDOT-FCP http://13.56.179.158/logs/85/570885/25/upstream-check/cinder-cDOT-FCP/843846c/ : FAILURE in 1h 02m 24s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":25},{"id":"67f0acd9819a4c04b8956e857abf583254d778f4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-06-12 19:54:50.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/85/570885/25/upstream-solidfire-check/cinder-solidfire-iSCSI/6e92b26/ : FAILURE in 4h 05m 58s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":25},{"id":"888c8bfcd407f071523cfc94c31a745865154c02","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-06-13 03:53:48.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/25/check/DellEMC_XtremIO_FC/e196841/DellEMC_XtremIO_FC/586 : SUCCESS in 51m 20s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/25/check/DellEMC_XtremIO_iSCSI/2debf44/DellEMC_XtremIO_iSCSI/544 : FAILURE in 2h 56m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":25},{"id":"fefa7d9f3601707ba4c482015b0b7b12520378c8","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-13 08:35:06.000000000","message":"Patch Set 25:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":25},{"id":"04e78ec4751501cb915261f8a9778976264e3de5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-13 09:50:20.000000000","message":"Patch Set 25:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-25 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":25},{"id":"c0dc298c02ff8a36d7245093f31d3248d2aa59fd","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-13 10:09:32.000000000","message":"Patch Set 25:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":25},{"id":"0377960bb36d90acd12898482517fc805aa44ad8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-13 11:35:31.000000000","message":"Patch Set 25:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-25 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":25},{"id":"8137cbc1cd7b589dbb2e00e4da9362c641d55f4a","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-06-13 11:55:20.000000000","message":"Patch Set 25:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":25},{"id":"08a3401b4c68c8611463ed8b77332ffe351f0968","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-13 13:17:04.000000000","message":"Patch Set 25:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-25 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":25},{"id":"3917b1ef984b9dab02e34aac72e1320135117e32","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-06-14 08:36:15.000000000","message":"Patch Set 25:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/85/570885/25/check/synology-dsm-cinder-iscsi/387465f : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":25},{"id":"fef1ca8536ed10ec79aa10cb7b93fbfc53e90bb5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-06-19 14:17:17.000000000","message":"Removed Code-Review-2 by Eric Harney \u003ceharney@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":25},{"id":"ac67cd7d530583759d2469dfd1900785f3e06938","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-06-27 00:22:57.000000000","message":"Patch Set 25: Code-Review+2\n\nLooks like comments have been addressed.  Thanks.","accounts_in_message":[],"_revision_number":25},{"id":"a25c908e2501afcfcc9c8720962acc0ce8aecb83","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-01 18:23:20.000000000","message":"Patch Set 25: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":25},{"id":"3ba1e11797d6488281ecb61227313990a173a6e6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-01 18:24:57.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"7c0cf8e5bd6b654c8bc37c1a4ca240ce21b75ff8","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-07-10 08:41:54.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"054ce4c87926b466cf532cf2bdb204f100c58f01","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-07-10 08:41:55.000000000","message":"Patch Set 25:\n\n(7 comments)","accounts_in_message":[],"_revision_number":25},{"id":"50268f882ef43e5278770770f41c431d84226dbd","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-10 08:42:10.000000000","message":"Patch Set 26:\n\nStart to Build","accounts_in_message":[],"_revision_number":26},{"id":"48d6c60ce701e3f0b3a88d88c1b5639254d6a9cc","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-10 08:42:13.000000000","message":"Patch Set 26:\n\nStart to Build","accounts_in_message":[],"_revision_number":26},{"id":"a37c47a2038fcadca4b92ecb81c81e8973afa740","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-07-10 08:42:16.000000000","message":"Patch Set 26:\n\nStart to Build","accounts_in_message":[],"_revision_number":26},{"id":"e6e3c220993c861cd969b27f2c5a6f4e106b5f75","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-10 09:10:38.000000000","message":"Patch Set 26:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1863 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1711 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1658 : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"9984fd5ebdbfb7f81c6fad3ebe2d21320c67313a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 09:37:36.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/26/check/dsvm-tempest-kvm/b0b90c9 : FAILURE in 41m 00s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/85/570885/26/check/dsvm-tempest-exe/c735ab1 : FAILURE in 35m 16s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":26},{"id":"67ac3c5af08c1af5e18273be57b9118bd35ceb98","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-10 10:17:29.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/85/570885/26 : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"5430f06593aa6ba25d7c7879448890abf1ec1a23","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-10 10:24:41.000000000","message":"Patch Set 26:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/85/570885/26/check/fujitsu-eternus-dx-iscsi/93abebf : SUCCESS in 1h 34m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/85/570885/26/check/fujitsu-eternus-dx-fc/f5d93e1 : SUCCESS in 1h 41m 53s","accounts_in_message":[],"_revision_number":26},{"id":"e03f56c5432d13b3a11853d09ccf390d3421601f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-10 10:29:50.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/85/570885/26/check-cinder/Cinder-ISER-ISCSI/8e2a7ea : SUCCESS in 1h 11m 12s\n- Cinder-ISER-LIO http://13.74.249.42/85/570885/26/check-cinder/Cinder-ISER-LIO/5152f20 : SUCCESS in 1h 06m 46s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":26},{"id":"ce8ea02dbe28b7396f03d7f41367c523596ea9c7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-10 10:30:37.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/85/570885/26/check/dsvm-tempest-infinibox-fc/10c3d83 : SUCCESS in 1h 34m 53s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/85/570885/26/check/dsvm-tempest-infinibox-iscsi/599bb1a : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":26},{"id":"5be43a101b9faa5ee78326f28c39f3e6e2c194de","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 10:45:40.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/85/570885/26/check/NEC-ISCSI/d5d97b1 : SUCCESS in 2h 02m 29s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/85/570885/26/check/NEC-FC/d7d2de5 : SUCCESS in 2h 01m 58s","accounts_in_message":[],"_revision_number":26},{"id":"b5d1ab71ab13494339446680b9643eef2378fdc5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-10 10:50:54.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/85/570885/26/check/DellEMC_Unity_FC/bb6e57c/DellEMC_Unity_FC/626 : SUCCESS in 1h 13m 04s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/26/check/DellEMC_Unity_iSCSI/8678409/DellEMC_Unity_iSCSI/596 : SUCCESS in 1h 24m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"d028b414388eada1bc8117be8e5fa219cf5d026b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-10 11:01:07.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/570885/26/check/dsvm-tempest-cisco-zonemanager-job/236d9d4/ : SUCCESS in 1h 15m 43s","accounts_in_message":[],"_revision_number":26},{"id":"6f874f1845c27bb00724e9db63975333c0caee78","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 11:12:35.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/26/check/DellEMC_XtremIO_FC/107d28e/DellEMC_XtremIO_FC/806 : SUCCESS in 1h 29m 01s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/26/check/DellEMC_XtremIO_iSCSI/5c4e6fe/DellEMC_XtremIO_iSCSI/753 : SUCCESS in 1h 30m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":26},{"id":"008094de77bd5594272d24a67ca31b72cca74dcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 11:12:51.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/570885/26/check/openstack-tox-lower-constraints/1468d1a/ : SUCCESS in 10m 17s\n- openstack-tox-pep8 http://logs.openstack.org/85/570885/26/check/openstack-tox-pep8/a24d221/ : SUCCESS in 7m 54s\n- openstack-tox-py27 http://logs.openstack.org/85/570885/26/check/openstack-tox-py27/7d66455/ : SUCCESS in 9m 16s\n- openstack-tox-py36 http://logs.openstack.org/85/570885/26/check/openstack-tox-py36/b13d09e/ : SUCCESS in 8m 14s\n- openstack-tox-py37 http://logs.openstack.org/85/570885/26/check/openstack-tox-py37/3e3d88e/ : SUCCESS in 8m 52s\n- openstack-tox-docs http://logs.openstack.org/85/570885/26/check/openstack-tox-docs/67ad16e/html/ : SUCCESS in 6m 56s\n- tempest-full http://logs.openstack.org/85/570885/26/check/tempest-full/febe043/ : SUCCESS in 1h 41m 25s\n- neutron-grenade http://logs.openstack.org/85/570885/26/check/neutron-grenade/12f4b0e/ : SUCCESS in 55m 15s\n- grenade-py3 http://logs.openstack.org/85/570885/26/check/grenade-py3/05f0993/ : SUCCESS in 58m 03s\n- tempest-full-py3 http://logs.openstack.org/85/570885/26/check/tempest-full-py3/57349b3/ : SUCCESS in 1h 27m 57s\n- build-openstack-releasenotes http://logs.openstack.org/85/570885/26/check/build-openstack-releasenotes/0017b9e/html/ : SUCCESS in 11m 39s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/85/570885/26/check/cinder-tempest-dsvm-lvm-lio-barbican/d05950c/ : SUCCESS in 1h 29m 31s\n- cinder-tox-bandit-baseline http://logs.openstack.org/85/570885/26/check/cinder-tox-bandit-baseline/107f70c/ : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/85/570885/26/check/openstack-tox-functional-py36/6d9c439/ : SUCCESS in 4m 34s\n- cinder-rally-task http://logs.openstack.org/85/570885/26/check/cinder-rally-task/e936aaf/ : SUCCESS in 37m 56s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/85/570885/26/check/openstack-tox-pylint/1a98d93/ : SUCCESS in 4m 53s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/85/570885/26/check/cinder-plugin-ceph-tempest/c03c927/ : SUCCESS in 1h 24m 11s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/85/570885/26/check/openstack-tox-functional/70bba0c/ : SUCCESS in 5m 09s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/85/570885/26/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/ae58dac/ : SUCCESS in 1h 02m 45s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/85/570885/26/check/legacy-tempest-dsvm-lvm-multibackend/b951847/ : SUCCESS in 1h 29m 27s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/85/570885/26/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/6b60f3b/ : SUCCESS in 1h 44m 48s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/85/570885/26/check/tempest-slow-py3/5c2553b/ : SUCCESS in 2h 18m 05s","accounts_in_message":[],"_revision_number":26},{"id":"4a0c59ba446716e89d9466f7a42cd29c2ff04ccd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 11:20:28.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/85/570885/26/check/DellEMC_VxFlexOS/e7a5115/DellEMC_VxFlexOS/403 : FAILURE in 7m 17s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":26},{"id":"06aef7aa909a13b0a18d4304790370e128a98b2a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-10 11:29:35.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/85/570885/26/check/dsvm-tempest-zadara-driver/4add8f2 : SUCCESS in 1h 41m 25s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"628cf47e842129dbac549f411cede1ed1371bf96","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-10 12:05:52.000000000","message":"Patch Set 26:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-85-570885-26/ns5_iscsi : SUCCESS in 46m 38s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-85-570885-26/ns5_nfs : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":26},{"id":"88b7f28cb5f7bded633b0504f2560b2c23211237","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-07-10 12:08:30.000000000","message":"Patch Set 26:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/85/2327 : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"70a0f1e24ecf11eec50d259e3727d3c67d0b95a4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-10 12:09:53.000000000","message":"Patch Set 26:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/26/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/08d425e : FAILURE in 51m 12s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/85/570885/26/check/3par-fc-driver-master-client-pip-ssa02-dsvm/cbf451b : FAILURE in 53m 23s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/85/570885/26/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/1a25021 : FAILURE in 50m 50s","accounts_in_message":[],"_revision_number":26},{"id":"a16168fb83809c498dd5ac56382cc43b382adab5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 12:26:05.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/85/570885/26/check/DellEMC_VNX_iSCSI/214f379/DellEMC_VNX_iSCSI/387 : FAILURE in 34m 16s\n- DellEMC_VNX_FC http://publiclogs.emc.com/85/570885/26/check/DellEMC_VNX_FC/b81e0bb/DellEMC_VNX_FC/395 : FAILURE in 1h 12m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":26},{"id":"65212fbee5c9ddcdedb2bd132d74517e588e0835","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 12:32:00.000000000","message":"Patch Set 26:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-85-570885-26-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"387eccf9cc1f500694a89ffbc08bc166205a66ef","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-07-10 12:52:37.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/85/570885/26/check/DellEMC_SC_FC/3341d61/DellEMC_SC_FC/247 : SUCCESS in 1h 09m 05s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/85/570885/26/check/DellEMC_SC_iSCSI/ad0912c/DellEMC_SC_iSCSI/230 : SUCCESS in 1h 39m 30s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":26},{"id":"b09ba2b8ad4636bd94492f6dafb8796610775390","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-10 13:02:34.000000000","message":"Patch Set 26:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-570885-26 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":26},{"id":"b38d957a8ab8cc51de54fece4201d36022481e44","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-07-10 13:28:59.000000000","message":"Patch Set 26:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/85/570885/26/Lenovo-FC : SUCCESS in 44:35.86\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/85/570885/26/Lenovo-iSCSI : SUCCESS in 42:02.91","accounts_in_message":[],"_revision_number":26},{"id":"c65f60f253070aa44e37710cf2ce2960d55459c6","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-07-10 13:29:08.000000000","message":"Patch Set 26: Code-Review+2\n\nLooks like comments have been addressed.  I am ok with this.","accounts_in_message":[],"_revision_number":26},{"id":"7a788d7376a31e7f150619bfefa0bdc4ca77f5ed","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-07-10 13:33:20.000000000","message":"Patch Set 26:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/85/570885/26/HPMSA-FC : SUCCESS in 45:30.58\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/85/570885/26/HPMSA-iSCSI : SUCCESS in 45:00.19","accounts_in_message":[],"_revision_number":26},{"id":"e1f8b6517399a4fa9f1e3bf573563c708c6ed36f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-10 15:10:38.000000000","message":"Patch Set 26:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/570885/26/iscsi/ : FAILURE in 13m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/570885/26/smb/ : FAILURE in 14m 17s","accounts_in_message":[],"_revision_number":26},{"id":"eae7b89e63bdcff543b60771175cc4512bd0addd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-10 15:23:48.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/85/570885/26/check/DellEMC_Unity_FC/1d2f471/DellEMC_Unity_FC/630 : SUCCESS in 1h 08m 11s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/85/570885/26/check/DellEMC_Unity_iSCSI/6cd3682/DellEMC_Unity_iSCSI/599 : SUCCESS in 1h 37m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"603f000bffb37c36b85ecb5e332085f5aa9988b5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 15:31:30.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/85/570885/26/check/DellEMC_XtremIO_FC/edcdb0a/DellEMC_XtremIO_FC/809 : SUCCESS in 1h 27m 36s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/85/570885/26/check/DellEMC_XtremIO_iSCSI/e300f33/DellEMC_XtremIO_iSCSI/756 : SUCCESS in 1h 29m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":26},{"id":"b697f863d729047635cadb3353528b5f55b7c688","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-10 20:37:01.000000000","message":"Patch Set 26:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/85/570885/26/check/tempest-dsvm-full-xenial/e8a7245/ : FAILURE in 1h 32m 27s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/85/570885/26/check/devstack-plugin-ceph-dsvm-xenial/ea71ce4/ : SUCCESS in 1h 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"bc477abf3f97937f2857d3fe19cb450faf19ab0d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-10 22:07:55.000000000","message":"Patch Set 26:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/85/570885/26/check-not-vote/ext-cinder-zuul/03122af : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":26},{"id":"6cbc8053e200fec87dbe4efbd4659fb126937e84","author":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"date":"2019-07-17 08:01:38.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"ea987839e6155463f33df373f3ea954e6172cb5b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-04 18:32:52.000000000","message":"Patch Set 26: Code-Review+2\n\nLooks good to me, but will wait a little on approving to give Eric a chance to take a look and make sure his concerns are addressed.","accounts_in_message":[],"_revision_number":26},{"id":"3fe9ac6abc7479f5decbbe17b3203caff1ccb771","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-09-03 07:27:31.000000000","message":"Patch Set 26:\n\nLooks like Eric removed his -1 (thanks for revisiting).\n\nI\u0027d expect Gerrit to add a corresponding comment, thought?","accounts_in_message":[],"_revision_number":26},{"id":"7dbc7d46ece3581946c62fcc287385431abbdda6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 22:06:04.000000000","message":"Patch Set 26: Workflow+1","accounts_in_message":[],"_revision_number":26},{"id":"b6c550d8ee3bf83ae303ed6b710ced9c19f9c3e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 22:06:31.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"4c2f147a81d949f3af53d89789f79cdaf0a442b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 00:08:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":26},{"id":"9e5da269f390ac4a96daa8722ae320e3cc935a7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 00:08:01.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb6e6af6591d4953b3edd3eba0783f9a : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acac6cf7c2b243698582c26858df19bb : SUCCESS in 9m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd929ee4139445dbbfb6e9e98ae31f01 : SUCCESS in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a32bda5eecfe46b0bb9c664bfa722e59 : SUCCESS in 9m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/936b583a591645799f10fe6e45172668 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bca303cf00e4dbc8321de01bb734d43 : SUCCESS in 16m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/27f95a960938480eadd3129f0114525d : SUCCESS in 1h 53m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5c365e4bf49b4b719944d78afba5837b : SUCCESS in 1h 12m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6eea0ae66cf44598ae7700544ec4f81b : SUCCESS in 1h 21m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/021aee7cc5f344bab8400587689aafbc : SUCCESS in 1h 25m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea535550d6ff4634bae67424ed235190 : SUCCESS in 18m 54s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf9e8789c46049258d9432f2b95a5c65 : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":26},{"id":"b810db46a3358fd675d8ef9da3d8a7929217262a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 00:11:25.000000000","message":"Patch Set 26:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3a0a43d4d2449a983abe94c1fd06eb : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"78f2101fc5a84e81132e6c2adc89cfe681018d77","revisions":{"f49bd95fbf0c81b96e333e011f1b4e614b34aa86":{"kind":"REWORK","_number":1,"created":"2018-05-29 10:19:37.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/1"}}},"commit":{"parents":[{"commit":"5dcf4f52ada36b98d08584cd31a55698006d9e35","subject":"Merge \"V3 jsonschema validation: scheduler_hints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dcf4f52ada36b98d08584cd31a55698006d9e35"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"subject":"[WIP] Adds _context field for backup snapshot creation","message":"[WIP] Adds _context field for backup snapshot creation\n\nAdds the _context field when constructing snapshots for\nbackups in the remotefs driver.\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\nCloses-Bug: #1744692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f49bd95fbf0c81b96e333e011f1b4e614b34aa86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f49bd95fbf0c81b96e333e011f1b4e614b34aa86"}]},"branch":"refs/heads/master"},"db846680428fef253d5d444fac8842a58d7508fc":{"kind":"REWORK","_number":2,"created":"2018-05-29 14:06:37.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/2"}}},"commit":{"parents":[{"commit":"5dcf4f52ada36b98d08584cd31a55698006d9e35","subject":"Merge \"V3 jsonschema validation: scheduler_hints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dcf4f52ada36b98d08584cd31a55698006d9e35"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 14:05:37.000000000","tz":120},"subject":"[WIP] Add _context to online snapshots in remotefs driver","message":"[WIP] Add _context to online snapshots in remotefs driver\n\nFetches _context from the volume data during online snapshots\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\nCloses-Bug: #1744692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db846680428fef253d5d444fac8842a58d7508fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db846680428fef253d5d444fac8842a58d7508fc"}]},"branch":"refs/heads/master"},"f4d2473d6118b641de1bd4d0eed20eee6c84a9a2":{"kind":"REWORK","_number":3,"created":"2018-05-30 16:45:56.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/3"}}},"commit":{"parents":[{"commit":"5dcf4f52ada36b98d08584cd31a55698006d9e35","subject":"Merge \"V3 jsonschema validation: scheduler_hints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dcf4f52ada36b98d08584cd31a55698006d9e35"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-30 16:43:56.000000000","tz":120},"subject":"[WIP] Add _context to online snapshots in remotefs driver","message":"[WIP] Add _context to online snapshots in remotefs driver\n\nFetches _context from obj_context property during online snapshots\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4d2473d6118b641de1bd4d0eed20eee6c84a9a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4d2473d6118b641de1bd4d0eed20eee6c84a9a2"}]},"branch":"refs/heads/master"},"53e7e112fb74b80c26338a2cfffa2b82836d41ec":{"kind":"REWORK","_number":4,"created":"2018-12-19 11:27:32.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/4"}}},"commit":{"parents":[{"commit":"203db865a6e02809e8baa2fed717c9d6e9b904d6","subject":"Merge \"Rename BackupNFSSwiftBasedTestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/203db865a6e02809e8baa2fed717c9d6e9b904d6"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-12-19 11:26:52.000000000","tz":60},"subject":"[WIP] Add _context to online snapshots in remotefs driver","message":"[WIP] Add _context to online snapshots in remotefs driver\n\nFetches _context from volume during online snapshots\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53e7e112fb74b80c26338a2cfffa2b82836d41ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53e7e112fb74b80c26338a2cfffa2b82836d41ec"}]},"branch":"refs/heads/master"},"571b7dec8b60282c82900be1e0290ef2e363beb2":{"kind":"REWORK","_number":5,"created":"2018-12-19 16:41:47.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/5"}}},"commit":{"parents":[{"commit":"203db865a6e02809e8baa2fed717c9d6e9b904d6","subject":"Merge \"Rename BackupNFSSwiftBasedTestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/203db865a6e02809e8baa2fed717c9d6e9b904d6"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-12-19 16:41:20.000000000","tz":60},"subject":"[WIP] Add _context to online snapshots in remotefs driver","message":"[WIP] Add _context to online snapshots in remotefs driver\n\nFetches _context from volume during online snapshots\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/571b7dec8b60282c82900be1e0290ef2e363beb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/571b7dec8b60282c82900be1e0290ef2e363beb2"}]},"branch":"refs/heads/master"},"aad657980acc4061f6f1a62ae31f587a477c9888":{"kind":"REWORK","_number":6,"created":"2018-12-19 16:46:13.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/6"}}},"commit":{"parents":[{"commit":"203db865a6e02809e8baa2fed717c9d6e9b904d6","subject":"Merge \"Rename BackupNFSSwiftBasedTestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/203db865a6e02809e8baa2fed717c9d6e9b904d6"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-12-19 16:46:00.000000000","tz":60},"subject":"[WIP] Add _context to online snapshots in remotefs driver","message":"[WIP] Add _context to online snapshots in remotefs driver\n\nFetches _context from volume during online snapshots\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aad657980acc4061f6f1a62ae31f587a477c9888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aad657980acc4061f6f1a62ae31f587a477c9888"}]},"branch":"refs/heads/master"},"a3ec4bd71663b15b54abc2d982eb0798d56826e4":{"kind":"REWORK","_number":7,"created":"2018-12-20 10:22:30.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/7"}}},"commit":{"parents":[{"commit":"203db865a6e02809e8baa2fed717c9d6e9b904d6","subject":"Merge \"Rename BackupNFSSwiftBasedTestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/203db865a6e02809e8baa2fed717c9d6e9b904d6"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-12-20 10:22:13.000000000","tz":60},"subject":"[WIP] Add _context to online snapshots in remotefs driver","message":"[WIP] Add _context to online snapshots in remotefs driver\n\nFetches _context from volume during online snapshots\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3ec4bd71663b15b54abc2d982eb0798d56826e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3ec4bd71663b15b54abc2d982eb0798d56826e4"}]},"branch":"refs/heads/master"},"fea036b5a69d7af02352744249582187264e1e9e":{"kind":"REWORK","_number":8,"created":"2019-01-10 13:30:48.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/8"}}},"commit":{"parents":[{"commit":"203db865a6e02809e8baa2fed717c9d6e9b904d6","subject":"Merge \"Rename BackupNFSSwiftBasedTestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/203db865a6e02809e8baa2fed717c9d6e9b904d6"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-10 13:29:03.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fea036b5a69d7af02352744249582187264e1e9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fea036b5a69d7af02352744249582187264e1e9e"}]},"branch":"refs/heads/master"},"9f2a2d2fd59ba7063a125c89f3ef1b5511aacece":{"kind":"REWORK","_number":9,"created":"2019-01-23 08:37:10.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/9"}}},"commit":{"parents":[{"commit":"af84029b0774a86bec610a28f41901a479c4e58b","subject":"Merge \"Update FusionStorage Cinder Driver by using REST API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af84029b0774a86bec610a28f41901a479c4e58b"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-23 08:36:42.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f2a2d2fd59ba7063a125c89f3ef1b5511aacece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f2a2d2fd59ba7063a125c89f3ef1b5511aacece"}]},"branch":"refs/heads/master"},"29c73ef8cddf3e952328c14302c72ec4a57768f6":{"kind":"REWORK","_number":10,"created":"2019-01-23 11:32:06.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/10"}}},"commit":{"parents":[{"commit":"af84029b0774a86bec610a28f41901a479c4e58b","subject":"Merge \"Update FusionStorage Cinder Driver by using REST API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af84029b0774a86bec610a28f41901a479c4e58b"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-23 11:31:54.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29c73ef8cddf3e952328c14302c72ec4a57768f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29c73ef8cddf3e952328c14302c72ec4a57768f6"}]},"branch":"refs/heads/master"},"4d9405a52265a618eead081dbc7e71875e45c6af":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-01-24 10:42:41.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/11"}}},"commit":{"parents":[{"commit":"cc52c7be06e88d5ac9c2d13e952967d405fa97f6","subject":"[WIP] Update volume provider_location during pull based snap deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc52c7be06e88d5ac9c2d13e952967d405fa97f6"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-24 10:41:00.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d9405a52265a618eead081dbc7e71875e45c6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d9405a52265a618eead081dbc7e71875e45c6af"}]},"branch":"refs/heads/master"},"758070a158d67a17db37f4c48504cc9c361a0b34":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-01-25 08:44:37.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/12"}}},"commit":{"parents":[{"commit":"22942cd4165b39a7e19486e167b0892801285df5","subject":"[WIP] Update volume provider_location during pull based snap deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22942cd4165b39a7e19486e167b0892801285df5"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-25 08:44:03.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/758070a158d67a17db37f4c48504cc9c361a0b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/758070a158d67a17db37f4c48504cc9c361a0b34"}]},"branch":"refs/heads/master"},"7b63810b8978f49e9d67e5b329965c1248670cd4":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-01-25 11:25:45.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/13"}}},"commit":{"parents":[{"commit":"a0de2e3079734fdde40144c48f8184e1559ffefa","subject":"[WIP] Update volume provider_location during pull based snap deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0de2e3079734fdde40144c48f8184e1559ffefa"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-25 11:25:26.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b63810b8978f49e9d67e5b329965c1248670cd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b63810b8978f49e9d67e5b329965c1248670cd4"}]},"branch":"refs/heads/master"},"471134040db7189a63e4c55fb974c488d33bfaf2":{"kind":"REWORK","_number":14,"created":"2019-01-30 14:38:03.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/14"}}},"commit":{"parents":[{"commit":"af84029b0774a86bec610a28f41901a479c4e58b","subject":"Merge \"Update FusionStorage Cinder Driver by using REST API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af84029b0774a86bec610a28f41901a479c4e58b"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-30 14:37:47.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/471134040db7189a63e4c55fb974c488d33bfaf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/471134040db7189a63e4c55fb974c488d33bfaf2"}]},"branch":"refs/heads/master"},"1719fdd6b9e7a4f5bbeee25eaef42d0165fdb508":{"kind":"REWORK","_number":15,"created":"2019-01-31 10:47:12.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/15"}}},"commit":{"parents":[{"commit":"af84029b0774a86bec610a28f41901a479c4e58b","subject":"Merge \"Update FusionStorage Cinder Driver by using REST API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af84029b0774a86bec610a28f41901a479c4e58b"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-31 10:46:38.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1719fdd6b9e7a4f5bbeee25eaef42d0165fdb508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1719fdd6b9e7a4f5bbeee25eaef42d0165fdb508"}]},"branch":"refs/heads/master"},"c0472ee542901eeeadb7d4190f35e5ac649ff3c0":{"kind":"REWORK","_number":16,"created":"2019-01-31 10:54:15.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/16"}}},"commit":{"parents":[{"commit":"af84029b0774a86bec610a28f41901a479c4e58b","subject":"Merge \"Update FusionStorage Cinder Driver by using REST API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af84029b0774a86bec610a28f41901a479c4e58b"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-01-31 10:54:05.000000000","tz":60},"subject":"[WIP] Add context to cloning snapshots in remotefs driver","message":"[WIP] Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0472ee542901eeeadb7d4190f35e5ac649ff3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0472ee542901eeeadb7d4190f35e5ac649ff3c0"}]},"branch":"refs/heads/master"},"e1b6782adf9d1499267ad3c2951d02e308a7c04d":{"kind":"REWORK","_number":17,"created":"2019-04-05 13:10:48.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/17"}}},"commit":{"parents":[{"commit":"1dec92ed8fed3fe07f880875b494a21213d95c53","subject":"Merge \"NetApp: Return all iSCSI targets-portals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1dec92ed8fed3fe07f880875b494a21213d95c53"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-04-05 13:09:50.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1b6782adf9d1499267ad3c2951d02e308a7c04d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1b6782adf9d1499267ad3c2951d02e308a7c04d"}]},"branch":"refs/heads/master"},"88a998030acb91be523fceaf70f54783dae83b2f":{"kind":"REWORK","_number":18,"created":"2019-04-12 10:37:14.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/18"}}},"commit":{"parents":[{"commit":"1dec92ed8fed3fe07f880875b494a21213d95c53","subject":"Merge \"NetApp: Return all iSCSI targets-portals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1dec92ed8fed3fe07f880875b494a21213d95c53"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-04-12 10:36:45.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88a998030acb91be523fceaf70f54783dae83b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88a998030acb91be523fceaf70f54783dae83b2f"}]},"branch":"refs/heads/master"},"9a8fa48d264f8d458f7648705874465184b9faae":{"kind":"REWORK","_number":19,"created":"2019-06-03 10:17:13.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/19"}}},"commit":{"parents":[{"commit":"b8a41b8d42737fdd9f86acd4c43d75a9397875f8","subject":"Merge \"move sheepdog exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8a41b8d42737fdd9f86acd4c43d75a9397875f8"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-06-03 10:16:31.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a8fa48d264f8d458f7648705874465184b9faae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a8fa48d264f8d458f7648705874465184b9faae"}]},"branch":"refs/heads/master"},"a8c31633a68c1f65d37b7738a3998a69c833e7f8":{"kind":"REWORK","_number":20,"created":"2019-06-05 12:55:27.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/20"}}},"commit":{"parents":[{"commit":"b8a41b8d42737fdd9f86acd4c43d75a9397875f8","subject":"Merge \"move sheepdog exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8a41b8d42737fdd9f86acd4c43d75a9397875f8"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-06-05 12:54:58.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8c31633a68c1f65d37b7738a3998a69c833e7f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8c31633a68c1f65d37b7738a3998a69c833e7f8"}]},"branch":"refs/heads/master"},"f1cf92d2b1abfa94461bf147e27824f356da24de":{"kind":"REWORK","_number":21,"created":"2019-06-05 15:32:56.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/21"}}},"commit":{"parents":[{"commit":"b8a41b8d42737fdd9f86acd4c43d75a9397875f8","subject":"Merge \"move sheepdog exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8a41b8d42737fdd9f86acd4c43d75a9397875f8"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-06-05 15:31:59.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\nThis affects the NFS, WindowsSMBFS, VZStorage and\nQuobyte drivers.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1cf92d2b1abfa94461bf147e27824f356da24de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1cf92d2b1abfa94461bf147e27824f356da24de"}]},"branch":"refs/heads/master"},"642087468f7a8e3e3442bd1e0c9a1b8a2c62e7a8":{"kind":"REWORK","_number":22,"created":"2019-06-06 10:18:16.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/22"}}},"commit":{"parents":[{"commit":"b8a41b8d42737fdd9f86acd4c43d75a9397875f8","subject":"Merge \"move sheepdog exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8a41b8d42737fdd9f86acd4c43d75a9397875f8"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-06-06 10:17:59.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\nThis affects the NFS, WindowsSMBFS, VZStorage and\nQuobyte drivers.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/642087468f7a8e3e3442bd1e0c9a1b8a2c62e7a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/642087468f7a8e3e3442bd1e0c9a1b8a2c62e7a8"}]},"branch":"refs/heads/master"},"33046411a3d7fbc30bac700886c6fd461642178e":{"kind":"REWORK","_number":23,"created":"2019-06-07 09:18:00.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/23"}}},"commit":{"parents":[{"commit":"b8a41b8d42737fdd9f86acd4c43d75a9397875f8","subject":"Merge \"move sheepdog exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8a41b8d42737fdd9f86acd4c43d75a9397875f8"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-06-07 09:17:42.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\nThis affects the NFS, WindowsSMBFS, VZStorage and\nQuobyte drivers.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33046411a3d7fbc30bac700886c6fd461642178e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33046411a3d7fbc30bac700886c6fd461642178e"}]},"branch":"refs/heads/master"},"a180073f89aa9c3227bb1875007697a8960b0b28":{"kind":"REWORK","_number":24,"created":"2019-06-11 07:51:55.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/24"}}},"commit":{"parents":[{"commit":"b8a41b8d42737fdd9f86acd4c43d75a9397875f8","subject":"Merge \"move sheepdog exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8a41b8d42737fdd9f86acd4c43d75a9397875f8"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-06-11 07:51:06.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\nThis affects the NFS, WindowsSMBFS, VZStorage and\nQuobyte drivers.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a180073f89aa9c3227bb1875007697a8960b0b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a180073f89aa9c3227bb1875007697a8960b0b28"}]},"branch":"refs/heads/master"},"6897a33a22072c294816aa71284da73ef7755015":{"kind":"REWORK","_number":25,"created":"2019-06-11 14:14:30.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/25"}}},"commit":{"parents":[{"commit":"b8a41b8d42737fdd9f86acd4c43d75a9397875f8","subject":"Merge \"move sheepdog exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8a41b8d42737fdd9f86acd4c43d75a9397875f8"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-06-11 14:14:02.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\nThis affects the NFS, WindowsSMBFS, VZStorage and\nQuobyte drivers.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6897a33a22072c294816aa71284da73ef7755015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6897a33a22072c294816aa71284da73ef7755015"}]},"branch":"refs/heads/master"},"78f2101fc5a84e81132e6c2adc89cfe681018d77":{"kind":"REWORK","_number":26,"created":"2019-07-10 08:41:54.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/85/570885/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/570885/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/570885/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/570885/26"}}},"commit":{"parents":[{"commit":"b8a41b8d42737fdd9f86acd4c43d75a9397875f8","subject":"Merge \"move sheepdog exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8a41b8d42737fdd9f86acd4c43d75a9397875f8"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2018-05-29 10:17:43.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2019-07-10 08:41:15.000000000","tz":120},"subject":"Add context to cloning snapshots in remotefs driver","message":"Add context to cloning snapshots in remotefs driver\n\nAdds context to _create_cloned_volume method in order\nto allow taking a snapshot during cloning. Also saves\nthe temp_snapshot in order to enable Nova to update the\nsnapshot state. Destroys the temp_snapshot after usage.\nThis affects the NFS, WindowsSMBFS, VZStorage and\nQuobyte drivers.\n\nCloses-Bug: #1744692\n\nChange-Id: I328a02d0f26e8c3d41ec18a0487da6fd20b39b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78f2101fc5a84e81132e6c2adc89cfe681018d77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78f2101fc5a84e81132e6c2adc89cfe681018d77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
