)]}'
{"id":"openstack%2Fcinder~585675","triplet_id":"openstack%2Fcinder~master~I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5","project":"openstack/cinder","branch":"master","topic":"bug/1783495","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-08-30 21:00:54.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-30 21:00:54.000000000","reason":"Change was submitted"},"23083":{"account":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"last_update":"2024-08-30 21:00:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5","subject":"Add API document for availability zone list","status":"MERGED","created":"2018-07-25 08:34:23.000000000","updated":"2024-09-01 23:58:17.000000000","submitted":"2024-08-30 21:00:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"585675-bug/1783495","meta_rev_id":"0595050c352d2f8ae4d9375bd23c977b59bda31d","_number":585675,"virtual_id_number":585675,"owner":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 21:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-08-28 14:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-30 18:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-30 18:25:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-25 09:04:31.000000000","updated_by":{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 05:39:34.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 05:52:02.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2018-07-26 05:56:56.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 06:04:23.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2018-07-26 06:14:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 06:15:19.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 06:51:18.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 07:14:40.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2018-07-26 07:39:33.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 07:47:33.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2018-07-26 07:48:26.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2018-07-26 07:57:27.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 08:56:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2018-07-26 09:55:40.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2018-07-26 11:05:46.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2018-07-26 14:43:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-11-10 08:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-10 08:46:13.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2018-11-10 09:06:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-10 09:36:16.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-10 09:37:46.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2018-11-10 09:40:15.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-10 09:47:37.000000000","updated_by":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"reviewer":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"state":"REVIEWER"},{"updated":"2018-11-10 09:50:13.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2018-11-10 10:17:06.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-10 10:19:48.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-10 10:56:12.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-10 11:24:57.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-11 13:48:20.000000000","updated_by":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"reviewer":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"state":"REVIEWER"},{"updated":"2018-11-18 03:50:00.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-24 01:31:45.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 08:05:23.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-14 08:39:01.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-14 08:40:33.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-14 09:29:04.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-14 09:49:28.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-08-14 09:53:07.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-14 13:12:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2024-08-14 17:53:15.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-08-14 23:46:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-15 14:03:47.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-08-28 19:35:22.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-30 18:25:27.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-09-01 23:58:17.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"}],"messages":[{"id":"ccd49801e7f087f42b1710ef02c42c7f32b72748","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-07-25 08:34:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"106f990b8874fc9c3a046d5395bfefec546163ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-25 08:49:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/75/585675/1/check/openstack-tox-pep8/8b8c572/ : SUCCESS in 7m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/75/585675/1/check/build-openstack-sphinx-docs/c1e18ef/html/ : SUCCESS in 7m 47s\n- build-openstack-api-ref http://logs.openstack.org/75/585675/1/check/build-openstack-api-ref/aa5d0e8/ : FAILURE in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a6feddb4368af661dd33c24e2cc6d250fef658fe","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-25 08:54:07.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/75/585675/1/check-not-vote/ext-cinder-zuul/6270a2c : FAILURE in 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1bf843c586832edf112b147c2b59222eacafe106","author":{"_account_id":12822,"name":"ProphetStor CI","email":"prophetstor.ci@prophetstor.com","username":"prophetstor.ci","tags":["SERVICE_USER"]},"date":"2018-07-25 09:04:31.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- prophetstor-dsvm-tempest-cinder-ci http://download.prophetstor.com/prophetstor_ci/585675/1/check/prophetstor-dsvm-tempest-cinder-ci/12761 : FAILURE in 29m 23s\n\nFor information of ProphetStor CI, see https://wiki.openstack.org/wiki/ThirdPartySystems/ProphetStor_CI","accounts_in_message":[],"_revision_number":1},{"id":"9df29fdc3f660017b9e5bb6650f8c6477ab6d927","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-07-25 09:26:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/585675/1/iscsi/ : SUCCESS in 52m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/585675/1/smb/ : SUCCESS in 42m 50s","accounts_in_message":[],"_revision_number":1},{"id":"aa0f4867682deef8537ec56aac951a6d79ac8288","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-07-25 09:37:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/75/585675/1/check/check-dsvm-tempest-vz7-exe-minimal/12ff899 : SUCCESS in 1h 03m 01s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"dd02419095edb1439715897bc52f0ae1dd799c3a","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-07-25 09:38:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93f534abb78cb2ada7958c0454d4902001349100","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-25 10:17:44.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/75/585675/2/check-not-vote/ext-cinder-zuul/e283df4 : FAILURE in 38m 39s","accounts_in_message":[],"_revision_number":2},{"id":"8139e4e2ca6420d2b7e153b4b80ce7cbd7dc8282","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-07-25 10:31:09.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/585675/2/iscsi/ : FAILURE in 52m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/585675/2/smb/ : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":2},{"id":"bc344cb829179f6eb84a6a2319ea39d2ddd151a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-25 10:33:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/75/585675/2/check/openstack-tox-pep8/b68010f/ : SUCCESS in 14m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/75/585675/2/check/build-openstack-sphinx-docs/183d5e7/html/ : SUCCESS in 7m 58s\n- build-openstack-api-ref http://logs.openstack.org/75/585675/2/check/build-openstack-api-ref/7ec7030/ : FAILURE in 10m 23s","accounts_in_message":[],"_revision_number":2},{"id":"4c75dd7b0eeac47aa9f96ab55d352fe9cfe450c5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-07-25 10:40:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/75/585675/2/check/check-dsvm-tempest-vz7-exe-minimal/c6f0e48 : SUCCESS in 1h 02m 05s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"5e97cbe08e06c0800de7f009da2161341f57c32f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-07-25 10:44:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/75/585675/2/check-cinder/Cinder-ISER-ISCSI/2ed1a30 : SUCCESS in 58m 40s\n- Cinder-ISER-LIO http://13.74.249.42/75/585675/2/check-cinder/Cinder-ISER-LIO/86eed3a : SUCCESS in 1h 02m 33s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"04f3c32e8cfce1599e3e645113edc26fa079fb60","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-07-25 10:50:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/75/585675/2/check/dsvm-tempest-infinibox-fc/2b853e4 : SUCCESS in 1h 02m 11s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/75/585675/2/check/dsvm-tempest-infinibox-iscsi/e12819d : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":2},{"id":"c41265c6eadaca022cae896709cc8a9fbd488bf2","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-07-25 11:29:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/11366 : FAILURE in 35m 42s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/7787 : FAILURE in 34m 37s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/9584 : FAILURE in 35m 40s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/8181 : FAILURE in 38m 28s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/9148 : FAILURE in 1h 07m 06s\n- flashsystem-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-fc/7825 : FAILURE in 1h 13m 53s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/424 : SUCCESS in 1h 41m 17s","accounts_in_message":[],"_revision_number":2},{"id":"2ce02b97354b290557c18896b2ed8058cb95a973","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-07-25 11:46:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/75/585675/2/check/dsvm-tempest-storpool/c6dbfc6 : SUCCESS in 1h 33m 39s","accounts_in_message":[],"_revision_number":2},{"id":"61a2fcea2ad5e0148c777aa6492fac8133a12539","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-07-25 11:48:46.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/75/585675/2/check/dsvm-veritas-access-cinder-driver/31183a0 : SUCCESS in 1h 48m 49s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/75/585675/2/check/dsvm-veritas-access-cinder-iscsi-driver/d6f8a6b : FAILURE in 2h 09m 00s","accounts_in_message":[],"_revision_number":2},{"id":"95e6ff8f61d5b31f36e2367e4fc7986fdcc46d83","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-07-25 12:39:45.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- EMC_Unity_FC http://publiclogs.emc.com/585675/2/EMC_Unity_FC/706 : SUCCESS in 1h 04m 57s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/585675/2/EMC_Unity_iSCSI/630 : SUCCESS in 1h 34m 24s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d8448767bc71013d99058fa1c20a959ffef73703","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-07-25 15:08:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cbb015c557c1e94eafc1775a43d3e9a1e07f5ad1","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-07-25 15:11:22.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/75/585675/2/check/NEC-ISCSI/89d7f58 : FAILURE in 48m 12s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/75/585675/2/check/NEC-FC/05970f8 : SUCCESS in 2h 48m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d3f754c4f4c3a88a9ef85f9ada0e432f30f5333b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-07-25 23:30:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/75/585675/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/d68198d : SUCCESS in 47m 47s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/75/585675/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/948d195 : SUCCESS in 1h 08m 39s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"fc597d10ab34f1d797d49704d43dbe5ad1286511","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-07-26 02:01:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1849bc972a556c3329881ff5f6d4c937673e6a92","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-07-26 02:02:32.000000000","message":"Patch Set 3:\n\n\u003e (5 comments)\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"796b7b8a5783abdc482e26fdfba91c1731052bf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-26 02:15:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/75/585675/3/check/openstack-tox-pep8/72d9f23/ : SUCCESS in 10m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/75/585675/3/check/build-openstack-sphinx-docs/206c669/html/ : SUCCESS in 9m 53s\n- build-openstack-api-ref http://logs.openstack.org/75/585675/3/check/build-openstack-api-ref/ec16cf0/ : FAILURE in 8m 11s","accounts_in_message":[],"_revision_number":3},{"id":"15caf315b0e90c0017c17b34353b33bac3edde41","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-07-26 02:53:23.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/585675/3/iscsi/ : SUCCESS in 51m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/585675/3/smb/ : FAILURE in 43m 11s","accounts_in_message":[],"_revision_number":3},{"id":"918b9c9844e7e0c170cd5392bec721fc91c0cc01","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-07-26 02:57:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/75/585675/3/check/check-dsvm-tempest-vz7-exe-minimal/a069bf2 : SUCCESS in 55m 44s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"0031c1cdeb70433f70396136d210e25e2720f32a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-07-26 03:05:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/75/585675/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/123c45a : SUCCESS in 55m 06s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/75/585675/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/c250ffa : SUCCESS in 1h 04m 04s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"e70bbde0be44a2d4bad858f2397f92b371661ef1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-07-26 03:43:26.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/75/585675/3/check/dsvm-tempest-infinibox-fc/1141134 : FAILURE in 30m 42s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/75/585675/3/check/dsvm-tempest-infinibox-iscsi/99bd959 : SUCCESS in 1h 41m 41s","accounts_in_message":[],"_revision_number":3},{"id":"d7abc8bf53916bcad16add2c289b178d73568aef","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-07-26 04:07:54.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/75/585675/3/check/dsvm-veritas-access-cinder-driver/b793ec6 : SUCCESS in 1h 16m 19s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/75/585675/3/check/dsvm-veritas-access-cinder-iscsi-driver/35e29f3 : FAILURE in 2h 04m 54s","accounts_in_message":[],"_revision_number":3},{"id":"16054600859682f4247747f569f30528891e119e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-07-26 04:28:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- EMC_Unity_FC http://publiclogs.emc.com/585675/3/EMC_Unity_FC/717 : SUCCESS in 1h 10m 14s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/585675/3/EMC_Unity_iSCSI/641 : SUCCESS in 1h 25m 18s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a11c27b3e3175f62e742739984c0a867d0d7a56f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-07-26 04:35:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/75/585675/3/check/dsvm-tempest-storpool/f6f2bf1 : SUCCESS in 1h 35m 06s","accounts_in_message":[],"_revision_number":3},{"id":"5eb4b0b6e67ec2fc4c9676618e0c4075b4ab363e","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-07-26 04:37:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"10c7bd80f39e8bced2b5365cb16703a704c788ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-26 04:47:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/75/585675/4/check/openstack-tox-pep8/c3f8a2c/ : SUCCESS in 7m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/75/585675/4/check/build-openstack-sphinx-docs/97e5734/html/ : SUCCESS in 6m 51s\n- build-openstack-api-ref http://logs.openstack.org/75/585675/4/check/build-openstack-api-ref/5202d82/ : FAILURE in 5m 55s","accounts_in_message":[],"_revision_number":4},{"id":"1f739afc4dd8474f3bda3fd608c34f26f4076b40","author":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"date":"2018-07-26 05:01:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b9784a740ad7d70570772295d3bf429856d3abc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-26 05:13:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/585675/5/check/openstack-tox-pep8/4296802/ : SUCCESS in 7m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/75/585675/5/check/build-openstack-sphinx-docs/892091b/html/ : SUCCESS in 6m 22s\n- build-openstack-api-ref http://logs.openstack.org/75/585675/5/check/build-openstack-api-ref/473f01f/html/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":5},{"id":"7fb4376858a1c8de72fba7c477245d4f024c0422","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-26 05:39:34.000000000","message":"Patch Set 5:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-75-585675-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"c42fcebf0ee76fe7c1a95a96ea7964fc759fdbb3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-07-26 05:52:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/585675/5/iscsi/ : SUCCESS in 50m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/585675/5/smb/ : SUCCESS in 43m 30s","accounts_in_message":[],"_revision_number":5},{"id":"141ef012a3ce8ea29eb9014f6298a2d79433517d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-07-26 05:56:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/75/585675/5/check/check-dsvm-tempest-vz7-exe-minimal/b0f9cda : SUCCESS in 55m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"0dcfdaf8329213e4e8d8aa1868daa5aa495a8552","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-07-26 05:57:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/75/585675/5/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/96394e8 : SUCCESS in 43m 49s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/75/585675/5/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/10fd0dd : SUCCESS in 55m 53s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"050eed92f4aee75d77378df4d779f814a914357b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-07-26 06:04:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/75/585675/5/check/dsvm-tempest-infinibox-fc/437f9f1 : SUCCESS in 52m 10s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/75/585675/5/check/dsvm-tempest-infinibox-iscsi/84a995c : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":5},{"id":"b53a2a99f197d4878c6c9a5be8adb7aa7c6f7f3d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-07-26 06:06:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/75/585675/5/check-cinder/Cinder-ISER-ISCSI/b3c1518 : SUCCESS in 59m 27s\n- Cinder-ISER-LIO http://13.74.249.42/75/585675/5/check-cinder/Cinder-ISER-LIO/372d727 : SUCCESS in 1h 02m 45s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"982b63541dba67fe7dbaf0e4767fe1832f06c6a6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-07-26 06:14:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/75/585675/5/upstream-check/cinder-cDOT-iSCSI/c496328/ : FAILURE in 2m 08s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/75/585675/5/upstream-check/cinder-cDOT-NFS/2021a82/ : SUCCESS in 58m 16s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/75/585675/5/upstream-check/cinder-cDOT-FCP/ed6a485/ : SUCCESS in 59m 51s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"d4bc8606b703cb5cffbea078d39a4d184d2f377d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-07-26 06:15:19.000000000","message":"Patch Set 5:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-75-585675-5/ns5_iscsi : SUCCESS in 38m 41s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-75-585675-5/ns_iscsi : SUCCESS in 44m 25s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-75-585675-5/ns5_nfs : SUCCESS in 48m 38s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-75-585675-5/ns_nfs : SUCCESS in 50m 25s","accounts_in_message":[],"_revision_number":5},{"id":"cf67b442fe30999b167ee14125aa9c6f3f63a385","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-07-26 06:51:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/75/585675/5/upstream-eseries-check/cinder-eseries-iSCSI/7621d83/ : SUCCESS in 1h 05m 16s (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/75/585675/5/upstream-eseries-check/cinder-eseries-FCP/77b5953/ : SUCCESS in 1h 09m 49s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":5},{"id":"7ca3ebea85253998a48786ccd477603f1fd71d75","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2018-07-26 07:14:40.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Dell_SC_FC http://publiclogs.emc.com/585675/5/Dell_SC_FC/509 : FAILURE in 1h 57m 41s\n- Dell_SC_iSCSI http://publiclogs.emc.com/585675/5/Dell_SC_iSCSI/559 : FAILURE in 2h 00m 47s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":5},{"id":"fdbcfa38c890b8b7484b75385b48ca1d01e105fd","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-07-26 07:16:21.000000000","message":"Patch Set 5:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/75/585675/5/check/dsvm-veritas-access-cinder-driver/1412126 : SUCCESS in 1h 17m 24s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/75/585675/5/check/dsvm-veritas-access-cinder-iscsi-driver/3181540 : FAILURE in 2h 12m 06s","accounts_in_message":[],"_revision_number":5},{"id":"bf854c9cf66a5113101eda607b07fe5275f75e5c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-07-26 07:39:33.000000000","message":"Patch Set 5:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-c8k01-dsvm http://54.201.44.218/75/585675/5/check/3par-iscsi-driver-master-client-pip-c8k01-dsvm/360ccf6 : SUCCESS in 1h 47m 23s\n- 3par-fc-driver-master-client-pip-c8k01-dsvm http://54.201.44.218/75/585675/5/check/3par-fc-driver-master-client-pip-c8k01-dsvm/3a3496c : SUCCESS in 2h 25m 50s\n- lefthand-iscsi-driver-master-client-pip-vsaftc126-dsvm http://54.201.44.218/75/585675/5/check/lefthand-iscsi-driver-master-client-pip-vsaftc126-dsvm/823f1fe : FAILURE in 2h 05m 14s","accounts_in_message":[],"_revision_number":5},{"id":"d3f43185414734b58c253147653feade7cc0ab32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-07-26 07:43:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/75/585675/5/check/dsvm-tempest-storpool/4d9e8ad : SUCCESS in 1h 34m 01s","accounts_in_message":[],"_revision_number":5},{"id":"406a69195332a8875a98d97063148f2e8699ef6e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-07-26 07:47:33.000000000","message":"Patch Set 5:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/75/585675/5/HPMSA-FC : SUCCESS in 33:35.51\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/75/585675/5/HPMSA-iSCSI : SUCCESS in 37:11.10","accounts_in_message":[],"_revision_number":5},{"id":"f1f181a01f9e39e21eccd30f26bdd6aa30532f20","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-07-26 07:48:26.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/11379 : FAILURE in 35m 26s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/7800 : FAILURE in 34m 22s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/9597 : FAILURE in 34m 56s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/8194 : FAILURE in 43m 32s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/9161 : FAILURE in 1h 03m 34s\n- flashsystem-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-fc/7838 : FAILURE in 1h 14m 34s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/436 : SUCCESS in 2h 24m 12s","accounts_in_message":[],"_revision_number":5},{"id":"ea99286036f0f59cd96d72470fa0b0e4c8b03e88","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-07-26 07:57:27.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://172.22.1.21/75/585675/5/check/synology-dsm-cinder-iscsi/b73a00a : FAILURE in 1h 39m 31s","accounts_in_message":[],"_revision_number":5},{"id":"f7f7d131a983c76a8e8e56ee93653534a8b1ba75","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-07-26 08:56:27.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- EMC_Unity_FC http://publiclogs.emc.com/585675/5/EMC_Unity_FC/721 : SUCCESS in 1h 05m 08s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/585675/5/EMC_Unity_iSCSI/644 : SUCCESS in 1h 32m 58s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e3362d77a19d94f42a58943d1e3d833b571f7473","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-07-26 09:55:40.000000000","message":"Patch Set 5:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/75/585675/5/Lenovo-FC : SUCCESS in 32:58.04\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/75/585675/5/Lenovo-iSCSI : SUCCESS in 38:08.48","accounts_in_message":[],"_revision_number":5},{"id":"96f088a39899a526d367a70bc4c1294f9e5eb99d","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-07-26 10:54:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/75/585675/5/check/NEC-ISCSI/2507c8c : SUCCESS in 1h 59m 07s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/75/585675/5/check/NEC-FC/b5098e4 : SUCCESS in 2h 05m 39s","accounts_in_message":[],"_revision_number":5},{"id":"b59cb52cb9f419cdbfe537f13e08feaa4ded15c0","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-07-26 11:05:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/75/585675/5/check/kaminario-dsvm-tempest-full-iscsi/2886e29 : SUCCESS in 1h 51m 14s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/75/585675/5/check/kaminario-dsvm-tempest-full-FC/62e5d9d : SUCCESS in 1h 48m 35s","accounts_in_message":[],"_revision_number":5},{"id":"0e5c66fb1e760a3cdca2cea950cd7457a691a12d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-07-26 14:43:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8dfbcf8a23fa3079b5f1fca44e26e72f7776d815","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-26 15:26:27.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/75/585675/5/check-not-vote/ext-cinder-zuul/1b98dab : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":5},{"id":"8ac7180acf87b08a8d3f6dfaac8635c4868f9df8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2018-07-30 07:42:52.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- EMC_VMAX_FC http://publiclogs.emc.com/585675/5/EMC_VMAX_FC/654 : SUCCESS in 7h 37m 53s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/585675/5/EMC_VMAX_iSCSI/632 : SUCCESS in 7h 19m 27s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":5},{"id":"2b5fa018a78ff881e3569af31a5850395ff1f74f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-11-10 08:30:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"add3b55573bcc5617875462f3d57c08d2b1e28c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-10 08:41:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/75/585675/6/check/build-openstack-api-ref/8c65729/ : FAILURE in 6m 45s\n- openstack-tox-pep8 http://logs.openstack.org/75/585675/6/check/openstack-tox-pep8/4526e7d/ : SUCCESS in 7m 20s\n- openstack-tox-docs http://logs.openstack.org/75/585675/6/check/openstack-tox-docs/36714f7/ : FAILURE in 5m 48s","accounts_in_message":[],"_revision_number":6},{"id":"1b0514144ff69ab0bb7704c408a7e018a7ae04e2","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-11-10 08:46:13.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/292 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/459 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"9bcb2acc72843648033fb4c67105f6f6118d1a61","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-11-10 09:06:37.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/75/585675/6/check-not-vote/ext-cinder-zuul/18bdae6 : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":6},{"id":"f61ed1aa62acb560901ac557376b61e613795dab","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-11-10 09:36:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/75/585675/6/check/dsvm-tempest-kvm/435f357 : SUCCESS in 1h 04m 59s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/75/585675/6/check/dsvm-tempest-exe/a4c19b5 : SUCCESS in 52m 03s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"62d6cd1bcb5438c8c7d11eeb1151668809e80bf0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-11-10 09:37:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/75/585675/6/check/dsvm-tempest-cisco-zonemanager-job/41516c2/ : FAILURE in 55m 47s","accounts_in_message":[],"_revision_number":6},{"id":"2b2a587a079fcddd7b8aacfdba083d84f0e33656","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-11-10 09:40:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/75/585675/6/check-cinder/Cinder-ISER-ISCSI/7cb55f7 : SUCCESS in 1h 04m 24s\n- Cinder-ISER-LIO http://13.74.249.42/75/585675/6/check-cinder/Cinder-ISER-LIO/0d0263a : SUCCESS in 1h 08m 51s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"bffe8f1642041b05fbc1b1e4f751b40cb6519c1b","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-11-10 09:47:37.000000000","message":"Patch Set 6:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/75/585675/6/check/dsvm-veritas-access-cinder-driver/2b575cd : FAILURE in 1h 16m 33s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/75/585675/6/check/dsvm-veritas-access-cinder-iscsi-driver/bf7eba8 : FAILURE in 21s","accounts_in_message":[],"_revision_number":6},{"id":"ecde8e0ee70a0b4a45877fe6cbed6f198c3fcd43","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-11-10 09:50:13.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/75/585675/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"43c185f2901c21ddd911b6be073aa9716790b8fa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-11-10 10:17:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/75/585675/6/check/dsvm-tempest-storpool/c2c8a84 : SUCCESS in 1h 46m 11s","accounts_in_message":[],"_revision_number":6},{"id":"c723d53fc09e49c2aa2c03b6d440ff34b8d40760","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-11-10 10:19:48.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/75/585675/6/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c4d576a : SUCCESS in 47m 03s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/75/585675/6/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/7664ac6 : FAILURE in 1h 49m 03s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"e45406f044c1a7743ed306fce182cbbac9cd4d2b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-11-10 10:56:12.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/75/585675/6/check/NEC-ISCSI/082a033 : FAILURE in 2h 24m 42s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/75/585675/6/check/NEC-FC/98f4856 : SUCCESS in 2h 23m 35s","accounts_in_message":[],"_revision_number":6},{"id":"8cd78fa833e9ae1c8f571d131002823ed17e155f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-11-10 11:24:57.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/585675/6/EMC_XtremIO_FC/716 : FAILURE in 1h 49m 59s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/585675/6/EMC_XtremIO_iSCSI/735 : SUCCESS in 1h 20m 38s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":6},{"id":"412d7ec5359ce45baf83147214f07f7563d1da8d","author":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"date":"2018-11-11 13:48:20.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027datacore: recheck\u0027. For information about DataCore CI see https://wiki.openstack.org/wiki/ThirdPartySystems/DataCore_CI\n\n- dsvm-tempest-datacore-iscsi-driver http://173.221.96.158/75/585675/6/check/dsvm-tempest-datacore-iscsi-driver/3f19d84 : FAILURE in 47m 45s\n- dsvm-tempest-datacore-fc-driver http://173.221.96.158/75/585675/6/check/dsvm-tempest-datacore-fc-driver/9b85ce4 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":6},{"id":"e53cf9b0e14f7b510f1e72bd26799309ff71f8df","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2018-11-18 03:50:00.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_FC http://publiclogs.emc.com/585675/6/EMC_VMAX_FC/438 : SUCCESS in 6h 33m 48s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/585675/6/EMC_VMAX_iSCSI/411 : FAILURE in 1m 21s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":6},{"id":"287babad7f0f084497e7662f89e0ed90d6905a00","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2018-11-24 01:31:45.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/75/585675/6/check/nimble-iscsi-driver-dsvm-volume/696bfd5 : FAILURE in 3h 00m 00s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/75/585675/6/check/nimble-fc-driver-dsvm-volume/933c7b6 : FAILURE in 2h 49m 44s","accounts_in_message":[],"_revision_number":6},{"id":"eef18413dad8ca018ff78abfb51fecb676d27866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-14 08:04:41.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"83dfc1b95ee015c71f1a777d6e91d7aa687ce75f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-14 08:05:23.000000000","message":"Patch Set 7:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":7},{"id":"6065091d8a17fec0b093fdf600005355d9649bce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-14 08:33:06.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/75/585675/7/check/DellEMC_Unity_FC/dd5db0e/DellEMC_Unity_FC/2403 : FAILURE in 19m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/7/check/DellEMC_Unity_iSCSI/5ecfe94/DellEMC_Unity_iSCSI/2408 : FAILURE in 28m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"e40b8a1491c311b98725480d276b270817881242","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-08-14 08:39:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/75/585675/7/ : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":7},{"id":"f384e8a56221c138b096b724413bfa76af356eca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-14 08:40:33.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/75/585675/7/check/DellEMC_PowerFlex_3.6/a8f3d02/DellEMC_PowerFlex_3.6/1888 : FAILURE in 30m 22s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/75/585675/7/check/DellEMC_PowerFlex_4.5/840b361/DellEMC_PowerFlex_4.5/767 : FAILURE in 35m 30s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/75/585675/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/d86b47c/DellEMC_PowerFlex_4.x_NVMeTCP/736 : FAILURE in 14m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"433a664642f426d050ff57633f759d7adafd7d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 08:42:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a886a720f7504e509c35511bc3c06afe\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eaa6c9faded94206819a8d319b575c79 : FAILURE in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc50a9bab6d4889b02ddaa1ee729939 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6d12a555044a99a737da4ef4459fb1 : SUCCESS in 10m 13s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1ab32b20b37541cbb161d267bacbb5cb : SUCCESS in 4m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/871ae032022a47b783ddfa95611884b5 : SUCCESS in 5m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2cb56d723e504f10ae64d59485341cec : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e34bddd9059b4311bc878a9dbf2856da : FAILURE in 37m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5d068c5a1b9541627d760cba977cb386111252d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-14 08:51:44.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d0c2ce5e12e44c3eeda5e388651f3f4cb590fc76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-14 08:51:55.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/280/","accounts_in_message":[],"_revision_number":8},{"id":"5ad5c066f6a809f1a1d37ebaa20d989b4e681ce5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-14 08:57:21.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/280/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/585675/8/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/585675/8/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"f5fc0b1a3e18876e90ccc75ca4a12b6babedabde","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-14 09:19:27.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_Unity_FC/4e18d40/DellEMC_Unity_FC/2404 : FAILURE in 19m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_Unity_iSCSI/8962696/DellEMC_Unity_iSCSI/2409 : FAILURE in 27m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"e910706dc51a81c527f96fbab9e1a176bdb8e06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 09:22:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d5e35f94364a01a9227036183618ca\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ce1d584519d248ee8dcc72b798b9c6b8 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed4392d7af04ab5b41923edbe9100f4 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea401d62a5d41089c0c847a38827ce5 : SUCCESS in 12m 24s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7642c30be555458d8869e29caa1c939d : SUCCESS in 7m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a8a34dff47214d00937bc37bbbaa1135 : SUCCESS in 6m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4dde46d8360542a88e41c46e38cd4c23 : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b65e9831f50f4c9b9748c776ef0d4bb7 : FAILURE in 29m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d39947ece937f35049137d07d95f75fa889faca5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-08-14 09:25:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/75/585675/8/ : RETRY_LIMIT in 9s","accounts_in_message":[],"_revision_number":8},{"id":"d03fddbbc3956aaa6e6a57a80104851b77c2dfee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-14 09:29:04.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_08_2024_09_23_23\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"9c7b2e8997f99973842402465600cfdec00c68f7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-14 09:36:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_PowerFlex_3.6/12ac662/DellEMC_PowerFlex_3.6/1889 : FAILURE in 30m 05s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_PowerFlex_4.5/41d39a8/DellEMC_PowerFlex_4.5/768 : FAILURE in 44m 07s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/c68fc6a/DellEMC_PowerFlex_4.x_NVMeTCP/737 : FAILURE in 13m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"b7a6f2418b1cbb4f0f37fb7eb32b772a322b5f90","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-14 09:49:28.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_PowerStore_iSCSI/48e6882/DellEMC_PowerStore_iSCSI/1601 : FAILURE in 27m 02s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_PowerStore_FC/4850130/DellEMC_PowerStore_FC/1797 : FAILURE in 20m 36s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_PowerStore_NFS/33b19ec/DellEMC_PowerStore_NFS/1294 : FAILURE in 24m 46s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_PowerStore_NVMeTCP/c3d5a96/DellEMC_PowerStore_NVMeTCP/1153 : FAILURE in 24m 54s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_PowerStore_NVMeFC/6fad1fc/DellEMC_PowerStore_NVMeFC/117 : FAILURE in 25m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"a2ebe8066a7f4808bd7f9430edd6065efa3717cc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-14 09:49:54.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_Unity_FC/99d2e4b/DellEMC_Unity_FC/2406 : FAILURE in 18m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/8/check/DellEMC_Unity_iSCSI/a5691ac/DellEMC_Unity_iSCSI/2411 : FAILURE in 26m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"da1a5668120fe928f9b9090077318ae8c247df38","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-14 09:53:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/585675/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/3364cda/ : RETRY_LIMIT in 6m 29s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/585675/8/upstream-check/cinder-tempest-plugin-ontap-nfs/1fb4918/ : RETRY_LIMIT in 7m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"9a98691d8b46dafa9f9dd59aeae8a4dc7b884cd3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-08-14 10:58:31.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"02f82fb84a54a6ff46c30e135cfa16ab09b347dc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-14 13:12:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"af618d112b2827af953b19c0f654bea7c949a2c1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-14 17:53:15.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/75/585675/8/20240814185001/NEC-V-FC : SUCCESS in 4h 06m 08s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/75/585675/8/20240814185001/NEC-V-ISCSI : FAILURE in 8h 03m 11s","accounts_in_message":[],"_revision_number":8},{"id":"c1b03c1be07a8a73890e0b07dfa10160806a1297","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-14 23:18:00.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/75/585675/8/20240814233010/NEC-ISCSI : SUCCESS in 5h 15m 59s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/75/585675/8/20240814233010/NEC-FC : SUCCESS in 8h 47m 46s","accounts_in_message":[],"_revision_number":8},{"id":"80eb78ff7c3bad1ca860a6ef69524e2107778d82","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-14 23:46:34.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/886/.","accounts_in_message":[],"_revision_number":8},{"id":"e17f2af8dcea6c1a37812b44d7a876b860d10fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-21 13:48:55.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"f2d068cdca8ca1e144491bfbf53790b2505ea854","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 13:49:05.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/356/","accounts_in_message":[],"_revision_number":9},{"id":"2f88f3c36a02dbc2a9d7f6e4bc0f90e298e8d39f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 13:52:30.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/356/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/585675/9/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/585675/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"5e55f4b9cedc478985cf29dde86837b13936c5b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:25:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4368a371d1e1422e98522cf4f6a5d0b8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8609531369c14a9d9a50ac1a23d8f464 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8517e452ff3549e893061e55b986cd69 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf38ba928d8d440a8ccd5e68f9916d7d : SUCCESS in 12m 28s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da7e562477d9485888f160ee85b4d51f : SUCCESS in 4m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2e73f206fda462390140571c800323f : SUCCESS in 6m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3a549e184e3a437a8517d11a7a9e1ee1 : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/15fc14fb2bb144bbbaf141032817f168 : FAILURE in 35m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ef0a0c9fbfc05908aac8df42ba3fbd758cb9fad1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-21 14:36:45.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_08_2024_14_31_12\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"9825077a6b679574b18847743e1dc7190215cf3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 14:38:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/362/.","accounts_in_message":[],"_revision_number":9},{"id":"2d56516740fb9599a95273e66a9b024def0a374c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-21 14:46:18.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/585675/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/89b3123/ : RETRY_LIMIT in 8m 56s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/585675/9/upstream-check/cinder-tempest-plugin-ontap-nfs/86df062/ : RETRY_LIMIT in 7m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"069c9c487f73016d33616a7fe849537077052b9b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-21 19:08:35.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_Unity_FC/efb0f3e/DellEMC_Unity_FC/2468 : SUCCESS in 1h 55m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_Unity_iSCSI/7370508/DellEMC_Unity_iSCSI/2472 : SUCCESS in 1h 58m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"c843ba610cc51847a63bfc940f35c4fa626b9a7c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-21 19:26:15.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_PowerStore_iSCSI/31330da/DellEMC_PowerStore_iSCSI/1627 : SUCCESS in 1h 17m 17s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_PowerStore_FC/e1b18ab/DellEMC_PowerStore_FC/1823 : SUCCESS in 1h 09m 01s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_PowerStore_NFS/719bf29/DellEMC_PowerStore_NFS/1319 : FAILURE in 1h 21m 59s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_PowerStore_NVMeTCP/f7b1490/DellEMC_PowerStore_NVMeTCP/1179 : SUCCESS in 1h 58m 41s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_PowerStore_NVMeFC/a83b993/DellEMC_PowerStore_NVMeFC/143 : SUCCESS in 1h 20m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"b7fbca17650aaefaa5be8297c7dfc7a2c7b79a04","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 21:55:27.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/955/.","accounts_in_message":[],"_revision_number":9},{"id":"aafc540decd88ae099b50de49bcba74e54c8bd23","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-08-23 00:22:53.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_PowerMAX_FC/eec1fcd/DellEMC_PowerMAX_FC/432 : SUCCESS in 5h 46m 13s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/9/check/DellEMC_PowerMAX_iSCSI/4391a51/DellEMC_PowerMAX_iSCSI/454 : FAILURE in 7h 25m 46s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":9},{"id":"540e40d4e086953dca3522e808b03f6a8d114e5a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-23 09:07:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/75/585675/9/20240823142546/NEC-V-FC : SUCCESS in 3h 39m 38s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/75/585675/9/20240823142546/NEC-V-ISCSI : SUCCESS in 3h 41m 29s","accounts_in_message":[],"_revision_number":9},{"id":"a6ad4705669543dbba1247a65421319e7fd0f760","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-28 14:57:48.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"59066f25ba1871339a02d2f666e1e15035d92e92","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-28 14:57:59.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/422/","accounts_in_message":[],"_revision_number":10},{"id":"5140554cacbc9b8959eadeb2f31e06adeacb702f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-28 15:01:36.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/585675/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/585675/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"d76d1d2cf1f8d324862bfc8b57b6a7925d85f8cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:38:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fbfaa14eed44c8aaf06189c7ee116d5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/90f581b6964f4d279eb5190dd772d318 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3f843499eb54f239172f4856e02db10 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b61c87a08934ea3854e4c1b8742ad2f : SUCCESS in 11m 52s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/208b948b3fed4a2da1ef1370dbcf97e0 : SUCCESS in 5m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f205765df8fa45dab441d42991e9a9d4 : SUCCESS in 5m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9aec8d581fcc433a9550a75d21b04702 : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eec44a604b784c9cb46a7057e82b2aef : FAILURE in 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8e56db23f39c958b8bcded188f5960eb96aa9371","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-28 15:50:02.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/585675/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/e55d5de/ : RETRY_LIMIT in 9m 45s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/585675/10/upstream-check/cinder-tempest-plugin-ontap-nfs/be3746e/ : RETRY_LIMIT in 4m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"b8333a3ed03715c1ae19b026735bba67d0cf2fe4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-28 18:04:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/427/.","accounts_in_message":[],"_revision_number":10},{"id":"378a0088c0232928a64d10afad13310051d8392d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-28 19:03:27.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_Unity_FC/fc9fe83/DellEMC_Unity_FC/None : EXCEPTION\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_Unity_iSCSI/b19acc9/DellEMC_Unity_iSCSI/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"fdaa1f7ea62574acfff930319c56a327e89abb9b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-28 19:35:22.000000000","message":"Patch Set 10:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/75/585675/10/check/fujitsu-eternus-dx-iscsi/42a61fe : SUCCESS in 2h 39m 03s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/75/585675/10/check/fujitsu-eternus-dx-fc/73ebed3 : SUCCESS in 2h 44m 59s","accounts_in_message":[],"_revision_number":10},{"id":"1fe1a71593cecf3cff025611730df02a0d013b26","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-28 22:57:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1015/.","accounts_in_message":[],"_revision_number":10},{"id":"34f344a4bb3cc03aa9984e949e0b98e2785fc5bd","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-29 00:04:49.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_PowerStore_iSCSI/e256167/DellEMC_PowerStore_iSCSI/1662 : SUCCESS in 1h 32m 17s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_PowerStore_FC/aff3861/DellEMC_PowerStore_FC/1858 : SUCCESS in 1h 08m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_PowerStore_NFS/5c91ecb/DellEMC_PowerStore_NFS/1355 : SUCCESS in 1h 10m 33s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_PowerStore_NVMeTCP/1a0bb9f/DellEMC_PowerStore_NVMeTCP/1215 : FAILURE in 2h 26m 01s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_PowerStore_NVMeFC/df5dce4/DellEMC_PowerStore_NVMeFC/179 : SUCCESS in 1h 16m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"83af44f037a5f5e98390800bb7f4b7c86040c65a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-30 03:04:34.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_PowerFlex_3.6/6d81779/DellEMC_PowerFlex_3.6/1989 : SUCCESS in 2h 39m 49s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_PowerFlex_4.x/2b7f6c6/DellEMC_PowerFlex_4.x/870 : SUCCESS in 2h 06m 44s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/75/585675/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/d598e4a/DellEMC_PowerFlex_4.x_NVMeTCP/842 : FAILURE in 13m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"3f6dee0ed3afb9e993a4edce5a9165baa92d8fb2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-30 10:36:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/75/585675/10/20240830153908/NEC-V-ISCSI : SUCCESS in 3h 44m 56s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/75/585675/10/20240830153908/NEC-V-FC : SUCCESS in 3h 57m 19s","accounts_in_message":[],"_revision_number":10},{"id":"6181f08de4e4e75deed4c1f33580a22420f42171","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-08-30 18:25:27.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2a5842f1c40451775952a8b541f8acd1b2446692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:26:15.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"f37c0b9563a309a6cb8004847eaed8fa20bd425f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 21:00:53.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989c46dbf847492683823fd1c030c00b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf74d9d424074baaa459281c722d2c9d : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3628e30a7e949c99e28fe0721022450 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec1269816714894b043770684603551 : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":10},{"id":"b122dd268aa24f010364f61a498da8aa5d0762a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 21:00:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"2e1a3374842172220c1f832e55c03d7eeb1ccab6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 21:02:16.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/389b9bcf2004401c8cf1dc781983e707\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/569cad9964c14a5ba3e10d04d0890a84 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b3ebc36ec640388279ad250ac9d7ae : SUCCESS in 54s","accounts_in_message":[],"_revision_number":10},{"id":"777c576ab018aa889e24b194e9fe5788356f419a","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-01 08:37:35.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/585675-10/HPMSA-iSCSI/ : SUCCESS in 3:17:22\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/585675-10/HPMSA-FC/ : FAILURE in 42:32.73","accounts_in_message":[],"_revision_number":10},{"id":"e3c61196cac5e55161d37625c2f11a98be922647","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-09-01 08:40:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/585675-10/Lenovo-iSCSI/ : SUCCESS in 2:22:28\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/585675-10/Lenovo-FC/ : FAILURE in 2:16:24","accounts_in_message":[],"_revision_number":10},{"id":"0595050c352d2f8ae4d9375bd23c977b59bda31d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-09-01 23:58:17.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/585675-10/Seagate-iSCSI/ : SUCCESS in 2:23:33\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/585675-10/Seagate-FC/ : SUCCESS in 2:17:27","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8a8082955293ec0241dcccb5c394c23d281b5479","revisions":{"59c5346044cd8dd7de84a558cae688d882b081a3":{"kind":"REWORK","_number":1,"created":"2018-07-25 08:34:23.000000000","uploader":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"ref":"refs/changes/75/585675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/1"}}},"commit":{"parents":[{"commit":"f236c48dd5919e6eca1fb524dd12c348e6a6a976","subject":"Merge \"Use Tempest scenario job to run all scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f236c48dd5919e6eca1fb524dd12c348e6a6a976"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:21:29.000000000","tz":480},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59c5346044cd8dd7de84a558cae688d882b081a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59c5346044cd8dd7de84a558cae688d882b081a3"}]},"branch":"refs/heads/master"},"a60dcd79998aff932eeed910b50a74210d61a68c":{"kind":"REWORK","_number":2,"created":"2018-07-25 09:38:26.000000000","uploader":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"ref":"refs/changes/75/585675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/2"}}},"commit":{"parents":[{"commit":"f236c48dd5919e6eca1fb524dd12c348e6a6a976","subject":"Merge \"Use Tempest scenario job to run all scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f236c48dd5919e6eca1fb524dd12c348e6a6a976"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 09:37:58.000000000","tz":480},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a60dcd79998aff932eeed910b50a74210d61a68c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a60dcd79998aff932eeed910b50a74210d61a68c"}]},"branch":"refs/heads/master"},"364f5005fc0d0a3219223c57f3684540c17d20be":{"kind":"REWORK","_number":3,"created":"2018-07-26 02:01:12.000000000","uploader":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"ref":"refs/changes/75/585675/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/3"}}},"commit":{"parents":[{"commit":"f236c48dd5919e6eca1fb524dd12c348e6a6a976","subject":"Merge \"Use Tempest scenario job to run all scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f236c48dd5919e6eca1fb524dd12c348e6a6a976"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-26 02:00:55.000000000","tz":480},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/364f5005fc0d0a3219223c57f3684540c17d20be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/364f5005fc0d0a3219223c57f3684540c17d20be"}]},"branch":"refs/heads/master"},"b3e8e08b583469de1783207a49e0db698e73edd5":{"kind":"REWORK","_number":4,"created":"2018-07-26 04:37:43.000000000","uploader":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"ref":"refs/changes/75/585675/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/4"}}},"commit":{"parents":[{"commit":"f236c48dd5919e6eca1fb524dd12c348e6a6a976","subject":"Merge \"Use Tempest scenario job to run all scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f236c48dd5919e6eca1fb524dd12c348e6a6a976"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-26 04:37:28.000000000","tz":480},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3e8e08b583469de1783207a49e0db698e73edd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3e8e08b583469de1783207a49e0db698e73edd5"}]},"branch":"refs/heads/master"},"a062e4e76cfbfd6b65338f29ea1340c5c639849d":{"kind":"REWORK","_number":5,"created":"2018-07-26 05:01:01.000000000","uploader":{"_account_id":23083,"name":"TommyLike","email":"tommylikehu@gmail.com","username":"TommyLike"},"ref":"refs/changes/75/585675/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/5"}}},"commit":{"parents":[{"commit":"f236c48dd5919e6eca1fb524dd12c348e6a6a976","subject":"Merge \"Use Tempest scenario job to run all scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f236c48dd5919e6eca1fb524dd12c348e6a6a976"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-26 04:59:20.000000000","tz":480},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a062e4e76cfbfd6b65338f29ea1340c5c639849d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a062e4e76cfbfd6b65338f29ea1340c5c639849d"}]},"branch":"refs/heads/master"},"c2e21c680eca5ef74eddbbeedd7bb9b4e705a0f6":{"kind":"REWORK","_number":6,"created":"2018-11-10 08:30:28.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/75/585675/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/6"}}},"commit":{"parents":[{"commit":"476ed701e2cc9bddba3d8cced7ba905ae42e61d6","subject":"Merge \"Add test coverage to manage existing flow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/476ed701e2cc9bddba3d8cced7ba905ae42e61d6"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2018-11-10 08:30:17.000000000","tz":-360},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2e21c680eca5ef74eddbbeedd7bb9b4e705a0f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2e21c680eca5ef74eddbbeedd7bb9b4e705a0f6"}]},"branch":"refs/heads/master"},"f3f7b0e2a14de07a43e553de24f495e172d0796b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-08-14 08:04:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/75/585675/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/7"}}},"commit":{"parents":[{"commit":"221a66940ebbba97ecfc4f8830cea50adb3a97a3","subject":"Merge \"Hitachi: Prevent to delete a LDEV assigned to multi objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/221a66940ebbba97ecfc4f8830cea50adb3a97a3"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-14 08:04:34.000000000","tz":120},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3f7b0e2a14de07a43e553de24f495e172d0796b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3f7b0e2a14de07a43e553de24f495e172d0796b"}]},"branch":"refs/heads/master"},"83b2be0ca19d8f3177f7864ff10a9bb22ef6b0c0":{"kind":"REWORK","_number":8,"created":"2024-08-14 08:51:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/75/585675/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/8"}}},"commit":{"parents":[{"commit":"221a66940ebbba97ecfc4f8830cea50adb3a97a3","subject":"Merge \"Hitachi: Prevent to delete a LDEV assigned to multi objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/221a66940ebbba97ecfc4f8830cea50adb3a97a3"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-14 08:51:43.000000000","tz":0},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83b2be0ca19d8f3177f7864ff10a9bb22ef6b0c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83b2be0ca19d8f3177f7864ff10a9bb22ef6b0c0"}]},"branch":"refs/heads/master"},"3e753c06bec5fa253b173ee05a2c4ade644cb2cf":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-08-21 13:48:55.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/75/585675/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/9"}}},"commit":{"parents":[{"commit":"1a0694a833e72ea9d31a4d37eef62ff0475ae32f","subject":"Merge \"Make Cinder reproducible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a0694a833e72ea9d31a4d37eef62ff0475ae32f"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-21 13:48:46.000000000","tz":120},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e753c06bec5fa253b173ee05a2c4ade644cb2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e753c06bec5fa253b173ee05a2c4ade644cb2cf"}]},"branch":"refs/heads/master"},"8a8082955293ec0241dcccb5c394c23d281b5479":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-08-28 14:57:48.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/75/585675/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/585675/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/585675/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/585675/10"}}},"commit":{"parents":[{"commit":"1185caa9a8f9e6fbc5cd2a476909ef3da3a6fd41","subject":"Merge \"add IPACL support to the Lightbits Cinder driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1185caa9a8f9e6fbc5cd2a476909ef3da3a6fd41"}]}],"author":{"name":"TommyLike","email":"tommylikehu@gmail.com","date":"2018-07-25 08:17:53.000000000","tz":480},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-28 14:57:48.000000000","tz":0},"subject":"Add API document for availability zone list","message":"Add API document for availability zone list\n\nAdd missing API document for list availability zone API.\n\nCloses-Bug: #1783495\nChange-Id: I04bf2063bd6d03fa509ff71e2fd66c1cb14e0fd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a8082955293ec0241dcccb5c394c23d281b5479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a8082955293ec0241dcccb5c394c23d281b5479"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
