)]}'
{"id":"openstack%2Fcinder~621465","triplet_id":"openstack%2Fcinder~master~Ie83dab4858729353865268231352033934dafdc2","project":"openstack/cinder","branch":"master","topic":"train-drivers","hashtags":[],"change_id":"Ie83dab4858729353865268231352033934dafdc2","subject":"Cinder RSD Driver","status":"MERGED","created":"2018-12-03 02:18:06.000000000","updated":"2019-06-14 11:55:07.000000000","submitted":"2019-06-13 07:50:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":182,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"621465-1560412207100-d9d96933","meta_rev_id":"6d9c98f551f47598068296d527af06ef72dde5b2","_number":621465,"virtual_id_number":621465,"owner":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-06-07 05:09:46.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 19:49:08.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 22:46:47.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2019-06-07 06:44:53.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 05:37:09.000000000","_account_id":16258,"name":"nikesh","email":"nike.niec@gmail.com","username":"nikem"},{"value":0,"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 20:36:36.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 21:58:48.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 21:58:10.000000000","_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},{"value":0,"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"date":"2019-06-07 05:46:36.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"date":"2019-06-13 03:56:30.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 08:02:45.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},{"value":0,"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30502,"name":"Jean-Baptiste Riaux","email":"jbriaux@kalray.eu"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"date":"2019-06-12 19:58:13.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-13 05:13:57.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 19:48:03.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":22450,"name":"qnap-ci","email":"openstack-ci@qnap.com","username":"qnap-ci"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"date":"2019-06-07 08:21:18.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 23:26:10.000000000","_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":23488,"name":"Helena","email":"helena.mcgough@intel.com","username":"hfm"},{"value":0,"date":"2019-06-07 05:20:48.000000000","_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2019-06-13 05:17:07.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"value":0,"date":"2019-06-06 23:40:39.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-06-13 07:50:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"date":"2019-06-13 07:08:17.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"value":0,"date":"2019-06-06 22:14:55.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-13 05:11:48.000000000","_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-13 06:09:35.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},{"value":0,"date":"2019-06-06 23:30:39.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"date":"2019-06-13 06:16:49.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-06-12 22:00:35.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"date":"2019-06-06 21:54:33.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-06-13 06:11:17.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2019-06-13 06:43:51.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2019-06-14 10:00:21.000000000","post_submit":true,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 22:01:33.000000000","_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-13 05:57:09.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 06:30:13.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 06:18:29.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-07 18:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16258,"name":"nikesh","email":"nike.niec@gmail.com","username":"nikem"},{"value":0,"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},{"value":0,"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},{"value":0,"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30502,"name":"Jean-Baptiste Riaux","email":"jbriaux@kalray.eu"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":22450,"name":"qnap-ci","email":"openstack-ci@qnap.com","username":"qnap-ci"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":23488,"name":"Helena","email":"helena.mcgough@intel.com","username":"hfm"},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 16:48:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16258,"name":"nikesh","email":"nike.niec@gmail.com","username":"nikem"},{"value":0,"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},{"value":0,"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},{"value":0,"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30502,"name":"Jean-Baptiste Riaux","email":"jbriaux@kalray.eu"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":22450,"name":"qnap-ci","email":"openstack-ci@qnap.com","username":"qnap-ci"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":23488,"name":"Helena","email":"helena.mcgough@intel.com","username":"hfm"},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 16:48:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-07 18:30:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16258,"name":"nikesh","email":"nike.niec@gmail.com","username":"nikem"},{"value":0,"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},{"value":0,"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"value":0,"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},{"value":0,"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30502,"name":"Jean-Baptiste Riaux","email":"jbriaux@kalray.eu"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":22450,"name":"qnap-ci","email":"openstack-ci@qnap.com","username":"qnap-ci"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2019-06-13 11:04:25.000000000","post_submit":true,"_account_id":23488,"name":"Helena","email":"helena.mcgough@intel.com","username":"hfm"},{"value":0,"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},{"_account_id":16258,"name":"nikesh","email":"nike.niec@gmail.com","username":"nikem"},{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22450,"name":"qnap-ci","email":"openstack-ci@qnap.com","username":"qnap-ci"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23488,"name":"Helena","email":"helena.mcgough@intel.com","username":"hfm"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},{"_account_id":30502,"name":"Jean-Baptiste Riaux","email":"jbriaux@kalray.eu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-04 05:11:00.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2018-12-14 10:11:47.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-12-17 08:54:33.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-01-04 17:06:08.000000000","updated_by":{"_account_id":22450,"name":"qnap-ci","email":"openstack-ci@qnap.com","username":"qnap-ci"},"reviewer":{"_account_id":22450,"name":"qnap-ci","email":"openstack-ci@qnap.com","username":"qnap-ci"},"state":"REVIEWER"},{"updated":"2019-01-09 20:36:08.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2019-01-10 10:49:35.000000000","updated_by":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"reviewer":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"state":"REVIEWER"},{"updated":"2019-01-10 11:22:12.000000000","updated_by":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"reviewer":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"state":"REVIEWER"},{"updated":"2019-01-10 20:48:25.000000000","updated_by":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"reviewer":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"state":"REVIEWER"},{"updated":"2019-01-17 12:12:01.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-01 05:53:24.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-03-09 00:09:31.000000000","updated_by":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"reviewer":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"state":"REVIEWER"},{"updated":"2019-03-10 07:27:29.000000000","updated_by":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"reviewer":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-04-26 01:16:57.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-02 21:07:48.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-05-02 21:14:16.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-05-02 21:19:41.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-27 10:09:19.000000000","updated_by":{"_account_id":30502,"name":"Jean-Baptiste Riaux","email":"jbriaux@kalray.eu"},"reviewer":{"_account_id":30502,"name":"Jean-Baptiste Riaux","email":"jbriaux@kalray.eu"},"state":"REVIEWER"},{"updated":"2019-05-27 13:09:12.000000000","updated_by":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"reviewer":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"state":"REVIEWER"},{"updated":"2019-05-28 01:03:13.000000000","updated_by":{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},"reviewer":{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},"state":"REVIEWER"},{"updated":"2019-05-29 09:35:31.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2019-05-29 13:41:32.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 05:39:00.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 15:45:34.000000000","updated_by":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"reviewer":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"state":"REVIEWER"},{"updated":"2019-05-30 16:43:10.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2019-05-30 18:40:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 14:32:13.000000000","updated_by":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"reviewer":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"state":"REVIEWER"},{"updated":"2019-06-06 21:54:33.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-06-06 21:58:10.000000000","updated_by":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"reviewer":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"state":"REVIEWER"},{"updated":"2019-06-06 22:01:33.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 22:14:55.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 23:26:10.000000000","updated_by":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 23:30:39.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-06-06 23:40:39.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 05:09:46.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 05:20:48.000000000","updated_by":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 05:46:36.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2019-06-07 06:18:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-06-07 06:30:13.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 06:44:53.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 08:02:45.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 08:21:18.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 18:30:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-07 21:58:48.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 05:37:09.000000000","updated_by":{"_account_id":16258,"name":"nikesh","email":"nike.niec@gmail.com","username":"nikem"},"reviewer":{"_account_id":16258,"name":"nikesh","email":"nike.niec@gmail.com","username":"nikem"},"state":"REVIEWER"},{"updated":"2019-06-12 16:48:58.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-06-12 19:48:03.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 19:49:08.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 19:58:13.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-06-12 20:36:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 22:00:35.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 22:46:47.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-06-13 03:56:30.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 05:11:48.000000000","updated_by":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 05:13:57.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 05:17:07.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 05:57:09.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2019-06-13 06:09:35.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-06-13 06:11:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 06:16:49.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-06-13 06:43:51.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2019-06-13 07:08:17.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-06-13 07:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 11:04:25.000000000","updated_by":{"_account_id":23488,"name":"Helena","email":"helena.mcgough@intel.com","username":"hfm"},"reviewer":{"_account_id":23488,"name":"Helena","email":"helena.mcgough@intel.com","username":"hfm"},"state":"REVIEWER"},{"updated":"2019-06-14 10:00:21.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b9abf0f40a6f16a641b067cf99c5cfda1737f31","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-03 02:18:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94254ad17a687d4b74603fbe8f76de87f4131477","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2018-12-03 02:18:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_FC http://publiclogs.emc.com/621465/1/EMC_VMAX_FC/605 : FAILURE in 27s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/621465/1/EMC_VMAX_iSCSI/581 : FAILURE in 25s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":1},{"id":"576242085d320a038b1c706162184c909083ddfa","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-03 02:35:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/3073028 : FAILURE in 16m 35s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/34486f0 : FAILURE in 16m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"1c5979660cfe7b67f19002d20bd8a20d8a9fd894","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-03 02:36:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/447 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/614 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"93250eee835f1fb94b61b9c7d1dbad66187a75f1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-03 03:07:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/1/check/fujitsu-eternus-dx-iscsi/7899342 : FAILURE in 40m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/1/check/fujitsu-eternus-dx-fc/a5769d6 : FAILURE in 44m 17s","accounts_in_message":[],"_revision_number":1},{"id":"281af8e74ddb2d75457795145a7bb4f00eb97322","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-03 03:09:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/1/iscsi/ : SUCCESS in 51m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/1/smb/ : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":1},{"id":"936418d14ef92165f83cf8606faf35e806f3226c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-03 03:21:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/1/check/tempest-dsvm-full-xenial/6ee6091/ : SUCCESS in 1h 03m 11s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/1/check/devstack-plugin-ceph-dsvm-xenial/0a55235/ : SUCCESS in 1h 01m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07b8d86f63048c0731b320e6c8ebecce8697d28e","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-03 04:01:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/1/check/dsvm-tempest-kvm/686a74d : FAILURE in 15m 54s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/1/check/dsvm-tempest-exe/20044f2 : FAILURE in 56m 16s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"77d672444097e2f6665d778e28d45c212214895a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-03 04:12:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/1/check-cinder/Cinder-ISER-ISCSI/84bc433 : SUCCESS in 1h 01m 23s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/1/check-cinder/Cinder-ISER-LIO/efd79fe : SUCCESS in 56m 45s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"459faf5716009daf2724f828bbe2b0b646a28eae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-03 04:23:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/1/check/openstack-tox-lower-constraints/aed6b46/ : SUCCESS in 8m 01s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/1/check/openstack-tox-pep8/53cc751/ : SUCCESS in 8m 38s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/1/check/openstack-tox-py27/15720ca/ : SUCCESS in 10m 11s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/1/check/openstack-tox-py35/89d1f05/ : SUCCESS in 8m 32s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/1/check/openstack-tox-py36/5149a4f/ : SUCCESS in 8m 25s\n- openstack-tox-docs http://logs.openstack.org/65/621465/1/check/openstack-tox-docs/aa5b1ab/html/ : SUCCESS in 6m 14s\n- tempest-full http://logs.openstack.org/65/621465/1/check/tempest-full/d3219b1/ : SUCCESS in 1h 21m 13s\n- neutron-grenade http://logs.openstack.org/65/621465/1/check/neutron-grenade/e24826c/ : SUCCESS in 56m 13s\n- grenade-py3 http://logs.openstack.org/65/621465/1/check/grenade-py3/a14eac2/ : SUCCESS in 1h 06m 14s\n- tempest-full-py3 http://logs.openstack.org/65/621465/1/check/tempest-full-py3/0587bf9/ : SUCCESS in 1h 22m 47s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/1/check/cinder-tempest-dsvm-lvm-lio-barbican/6fbfa5e/ : SUCCESS in 1h 57m 55s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/1/check/cinder-tox-bandit-baseline/6e9f8b1/ : SUCCESS in 6m 48s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/1/check/nova-multiattach/0da2c59/ : SUCCESS in 53m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/1/check/openstack-tox-functional-py35/01a67b0/ : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/1/check/openstack-tox-functional-py36/a24cfe2/ : SUCCESS in 7m 07s\n- cinder-rally-task http://logs.openstack.org/65/621465/1/check/cinder-rally-task/c6a5396/ : SUCCESS in 41m 55s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/1/check/openstack-tox-pylint/a0a2972/ : FAILURE in 4m 06s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/1/check/devstack-plugin-ceph-tempest/7776fee/ : SUCCESS in 1h 59m 29s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/1/check/openstack-tox-functional/0a2a00e/ : SUCCESS in 4m 33s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/1/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/14c4980/ : SUCCESS in 57m 48s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/1/check/legacy-tempest-dsvm-lvm-multibackend/526613b/ : SUCCESS in 1h 08m 17s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/1/check/legacy-tempest-dsvm-full-drbd-devstack/c3956c7/ : SUCCESS in 1h 28m 18s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/1/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/112042f/ : SUCCESS in 1h 17m 23s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/1/check/tempest-slow/22dfd11/ : SUCCESS in 1h 49m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e9002c4390a8987ea07c86e39473fe37a491d1a9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-03 04:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/1/check/dsvm-tempest-cisco-zonemanager-job/a12d5d5/ : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b8b0c23690ea0d9613ffdabcc041ee6372b7976e","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-03 05:00:03.000000000","message":"Patch Set 1:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-1/nedge_iscsi : FAILURE in 25m 34s","accounts_in_message":[],"_revision_number":1},{"id":"59a64a0e966834592cdeaa855e03c9a56381e41b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-03 05:29:55.000000000","message":"Patch Set 1:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-1/ns5_iscsi : SUCCESS in 39m 14s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-1/ns5_nfs : SUCCESS in 41m 7s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-1/ns_nfs : SUCCESS in 65m 27s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-1/ns_iscsi : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c51afbc568ab5ab78734ce833b5244b11f400136","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-03 05:45:16.000000000","message":"Patch Set 1:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"8a77f80255787c8f883986cfaf557b321b6479f2","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-03 05:54:32.000000000","message":"Patch Set 1:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-1-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2241105ea3e5cda13163120eb4aca86bfbe92831","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-03 05:56:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/1/check/NEC-ISCSI/252c6f8 : SUCCESS in 1h 56m 21s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/1/check/NEC-FC/eda093b : SUCCESS in 1h 59m 16s","accounts_in_message":[],"_revision_number":1},{"id":"91fbb31b2e0856683f461811c60b252eb6870b69","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-03 06:02:02.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/1/check/dsvm-veritas-access-cinder-driver/968ef0c : SUCCESS in 1h 27m 52s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/1/check/dsvm-veritas-access-cinder-iscsi-driver/b1db3f5 : FAILURE in 2h 22m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b98702e110f93e400ef554c979d8a172305f3890","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-03 06:54:05.000000000","message":"Patch Set 1:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/1/Lenovo-FC : SUCCESS in 35:56.75\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/1/Lenovo-iSCSI : SUCCESS in 36:52.17","accounts_in_message":[],"_revision_number":1},{"id":"59256ad55891482665ba945f01f6d5df6236fdd4","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-03 07:06:55.000000000","message":"Patch Set 1:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/1/HPMSA-FC : SUCCESS in 37:55.91\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/1/HPMSA-iSCSI : SUCCESS in 37:25.69","accounts_in_message":[],"_revision_number":1},{"id":"02a987562c85cf0a40d72d96a83ca2c0dfe5c886","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-03 07:18:07.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/1/check/synology-dsm-cinder-iscsi/87d6eef : FAILURE in 44m 08s","accounts_in_message":[],"_revision_number":1},{"id":"82bd97c9cdf92ad8d5347a852575724312663af8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-03 07:48:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/1/check/dsvm-tempest-zadara-driver/894216f : FAILURE in 1h 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e27c12428c4c28f7722f21310f41c2579a40a2f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-03 08:44:12.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/1/EMC_VxFlexOS/535 : SUCCESS in 1h 49m 06s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"65b66a669c5060f3278d0dc19b56cba9c9f3c9f9","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-03 08:53:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/1/check/kaminario-dsvm-tempest-full-iscsi/ec3edb0 : SUCCESS in 1h 48m 05s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/1/check/kaminario-dsvm-tempest-full-FC/586fdcb : SUCCESS in 1h 48m 04s","accounts_in_message":[],"_revision_number":1},{"id":"aa24c917a04e9c66a5291f66ee505eeb076ffebe","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2018-12-03 09:01:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_FC http://publiclogs.emc.com/621465/1/EMC_VMAX_FC/609 : FAILURE in 9s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/621465/1/EMC_VMAX_iSCSI/584 : FAILURE in 29s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":1},{"id":"0d638f1dd12fff7e3c1ea2f8a79b03f600ef840a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2018-12-03 09:09:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(18 comments)\n\nThanks for the contribution.\nPlease find some corrections inline.\nAlso name functions according to the guide [1] to maintain consistency with other drivers.\n\n[1] https://docs.openstack.org/cinder/latest/contributor/drivers.html#driver-development-documentations","accounts_in_message":[],"_revision_number":1},{"id":"8bc6ca1bb01ebf1db51ede3c66fc8eb415eb3dd9","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2018-12-03 09:21:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/621465/1/Dell_SC_FC/437 : SUCCESS in 2h 56m 28s\n- Dell_SC_iSCSI http://publiclogs.emc.com/621465/1/Dell_SC_iSCSI/429 : SUCCESS in 3h 00m 21s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":1},{"id":"7f699be2c49fec757f7d04856d4c1d093b864170","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-03 13:56:32.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0aadfd0 : FAILURE in 8m 11s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/1/check/3par-fc-driver-master-client-pip-ssa02-dsvm/07a1457 : SUCCESS in 1h 45m 47s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/1/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/0cbc8c0 : SUCCESS in 2h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a77aaef32a22b109719fffae0fb1a77fdfca6661","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-04 02:17:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b9a3dd6e4327fc0db1f7dea341ab97a27401034","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-04 02:28:36.000000000","message":"Patch Set 2:\n\n(18 comments)\n\nThanks Rajat for your reviews!\n\nAll the driver functions (and their names) inherit the base driver class, could you point out examples that don\u0027t consist with other drivers?\n\nOther comments are all replied inline.","accounts_in_message":[],"_revision_number":2},{"id":"57e83fc0d676a55bcfb8fa9c0b514a4fa4be4df8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-04 02:33:12.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/94b8256 : FAILURE in 15m 27s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/34ecda6 : FAILURE in 15m 15s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"1b2265a38c86a65121f26ba6240fc0eb1b768e02","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-04 02:37:06.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/631 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/464 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"53d7e148d034a125615268490d586bebbd1ecb1c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-04 02:53:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/2/check/fujitsu-eternus-dx-iscsi/f458a1c : FAILURE in 29m 33s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/2/check/fujitsu-eternus-dx-fc/141349b : FAILURE in 32m 01s","accounts_in_message":[],"_revision_number":2},{"id":"9e4155995e9738eb737d10a15f7df1708e81bcdb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-04 03:07:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/2/iscsi/ : SUCCESS in 49m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/2/smb/ : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4c47d6281271dcf80b620d1148e50ce4f3d7c77c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-04 03:20:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/2/check/dsvm-tempest-kvm/59e5749 : SUCCESS in 52m 28s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/2/check/dsvm-tempest-exe/badaaa9 : SUCCESS in 55m 52s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"843e4cf7a97689b17c497488922f717eb9c01d5b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-04 03:28:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/2/check/dsvm-tempest-cisco-zonemanager-job/86f54e5/ : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":2},{"id":"4348a6b5e8ec7fe0ab24dcd1934e5b94f5446b09","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-04 03:42:18.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/2/check/tempest-dsvm-full-xenial/3d220c8/ : SUCCESS in 1h 09m 52s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/2/check/devstack-plugin-ceph-dsvm-xenial/4a63842/ : SUCCESS in 1h 24m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e77165a2dc57f9afa986448d96201ef07a99a3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-04 04:28:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/2/check/openstack-tox-lower-constraints/a86cebf/ : SUCCESS in 10m 35s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/2/check/openstack-tox-pep8/c69e51d/ : SUCCESS in 7m 07s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/2/check/openstack-tox-py27/4eca38c/ : SUCCESS in 10m 36s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/2/check/openstack-tox-py35/5771134/ : SUCCESS in 8m 24s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/2/check/openstack-tox-py36/12ad012/ : SUCCESS in 7m 51s\n- openstack-tox-docs http://logs.openstack.org/65/621465/2/check/openstack-tox-docs/1b3271f/html/ : SUCCESS in 7m 48s\n- tempest-full http://logs.openstack.org/65/621465/2/check/tempest-full/dea3055/ : SUCCESS in 1h 22m 17s\n- neutron-grenade http://logs.openstack.org/65/621465/2/check/neutron-grenade/3458216/ : SUCCESS in 59m 32s\n- grenade-py3 http://logs.openstack.org/65/621465/2/check/grenade-py3/8243640/ : SUCCESS in 50m 13s\n- tempest-full-py3 http://logs.openstack.org/65/621465/2/check/tempest-full-py3/65e6670/ : SUCCESS in 1h 30m 15s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/2/check/cinder-tempest-dsvm-lvm-lio-barbican/54df24f/ : SUCCESS in 2h 00m 32s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/2/check/cinder-tox-bandit-baseline/199b755/ : SUCCESS in 6m 29s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/2/check/nova-multiattach/5e161f7/ : SUCCESS in 1h 08m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/2/check/openstack-tox-functional-py35/b218c8b/ : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/2/check/openstack-tox-functional-py36/8bee06b/ : SUCCESS in 6m 13s\n- cinder-rally-task http://logs.openstack.org/65/621465/2/check/cinder-rally-task/7809445/ : SUCCESS in 41m 03s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/2/check/openstack-tox-pylint/e186360/ : FAILURE in 5m 32s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/2/check/devstack-plugin-ceph-tempest/f3cf551/ : SUCCESS in 1h 38m 06s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/2/check/openstack-tox-functional/0532847/ : SUCCESS in 4m 45s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/2/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/6c8718d/ : SUCCESS in 1h 08m 35s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/2/check/legacy-tempest-dsvm-lvm-multibackend/79532d1/ : SUCCESS in 1h 14m 31s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/2/check/legacy-tempest-dsvm-full-drbd-devstack/6c1e593/ : SUCCESS in 1h 09m 51s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/2/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/6015752/ : SUCCESS in 1h 39m 43s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/2/check/tempest-slow/00d3bda/ : SUCCESS in 1h 35m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6119b17afc2745d216e07ad183c4f5694c8543fe","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-04 04:33:07.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/2/check/synology-dsm-cinder-iscsi/17cd5df : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"7ead916e898e384a608e082732a35bc3dd9c9066","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-04 04:34:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/2/EMC_VxFlexOS/545 : FAILURE in 5m 51s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"4bd722f2897c0c69448ba9aaae1276a47c74d120","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-04 04:39:21.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"8220c555ae018d393f2d9b431a6d71c30f0cddf2","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-04 04:42:45.000000000","message":"Patch Set 2:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-2/nedge_iscsi : FAILURE in 11m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9b95843c9aa19043c38f991823eb8b7af62445aa","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-04 04:42:48.000000000","message":"Patch Set 2:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-2/ns5_iscsi : FAILURE in 11m 43s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-2/ns_iscsi : FAILURE in 10m 54s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-2/ns5_nfs : FAILURE in 11m 19s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-2/ns_nfs : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8101123c382adca18e2008b03dfe3a332c16b030","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-04 04:54:49.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/2/check/kaminario-dsvm-tempest-full-iscsi/f9c7dce : FAILURE in 25m 05s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/2/check/kaminario-dsvm-tempest-full-FC/41b86e2 : FAILURE in 24m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8d3e8e1c4b695db4be39ff5898c858136afb5a77","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-04 04:57:39.000000000","message":"Patch Set 2:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-2-iscsi : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"53dd496078e31e28cca1690e9a8b22d369f01137","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2018-12-04 05:11:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nSome minor NITs, as i can see its a WIP but as a reminder, releasenote must be added too.","accounts_in_message":[],"_revision_number":2},{"id":"e17abaca5ac1ebda5eb2ad6f0086a98009e16ae9","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-04 05:52:21.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/2/check/dsvm-veritas-access-cinder-driver/837f19c : FAILURE in 1h 34m 55s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/2/check/dsvm-veritas-access-cinder-iscsi-driver/616f954 : FAILURE in 2h 19m 11s","accounts_in_message":[],"_revision_number":2},{"id":"7a9f30bc765a992bab446c17efdbd19eea2b02cd","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-04 07:05:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8582c7437fb22b7a6b5b218a174717a25a772dd","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-04 07:07:08.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nThanks for reminding, Rajat.\n\nAdded releasenot to the TODO list in the commit message.","accounts_in_message":[],"_revision_number":3},{"id":"a5a58a5a6c9d6d86f5cde55fc57d90f87766951c","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-04 07:21:28.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/633 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/466 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"8d732a40ea8535df1985ff89d3e9ce8bd3e88b3d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-04 07:28:13.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/3c63818 : FAILURE in 12m 28s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/d375e19 : FAILURE in 22m 13s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"5211fc19f24ffd28e7165299389cf528d493b749","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-04 08:06:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/3/iscsi/ : SUCCESS in 59m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/3/smb/ : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":3},{"id":"bb5c3dc2c3af6773422154dd33401d869b7cae5c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-04 08:16:19.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/3/check/tempest-dsvm-full-xenial/801f54f/ : SUCCESS in 1h 10m 21s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/3/check/devstack-plugin-ceph-dsvm-xenial/6250faf/ : SUCCESS in 1h 08m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9da278ba1f693b80a225a9b4be59a60c65e82fce","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-04 08:22:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/3/check/dsvm-tempest-kvm/103e9b0 : SUCCESS in 1h 15m 17s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/3/check/dsvm-tempest-exe/42cf5ab : FAILURE in 49m 35s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"d96c20f8cc5bba4c2a7a818cf3208afccbec900f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-04 09:12:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/3/check/fujitsu-eternus-dx-iscsi/8c51f84 : SUCCESS in 2h 04m 23s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/3/check/fujitsu-eternus-dx-fc/47b6393 : FAILURE in 47m 16s","accounts_in_message":[],"_revision_number":3},{"id":"72b394f179668fdc0a3a8569c8829c992e21e11c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-04 09:17:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/3/check/openstack-tox-lower-constraints/02c9605/ : SUCCESS in 9m 10s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/3/check/openstack-tox-pep8/39cec71/ : SUCCESS in 8m 53s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/3/check/openstack-tox-py27/2ff79f0/ : SUCCESS in 11m 10s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/3/check/openstack-tox-py35/32bac47/ : SUCCESS in 10m 04s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/3/check/openstack-tox-py36/6f40b67/ : SUCCESS in 7m 43s\n- openstack-tox-docs http://logs.openstack.org/65/621465/3/check/openstack-tox-docs/938f9f0/html/ : SUCCESS in 7m 11s\n- tempest-full http://logs.openstack.org/65/621465/3/check/tempest-full/ba32c78/ : SUCCESS in 1h 20m 22s\n- neutron-grenade http://logs.openstack.org/65/621465/3/check/neutron-grenade/e1de2da/ : SUCCESS in 59m 16s\n- grenade-py3 http://logs.openstack.org/65/621465/3/check/grenade-py3/0421601/ : SUCCESS in 55m 35s\n- tempest-full-py3 http://logs.openstack.org/65/621465/3/check/tempest-full-py3/fae4b8d/ : SUCCESS in 1h 26m 17s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/3/check/cinder-tempest-dsvm-lvm-lio-barbican/79b673a/ : SUCCESS in 1h 27m 04s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/3/check/cinder-tox-bandit-baseline/8b37015/ : SUCCESS in 7m 02s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/3/check/nova-multiattach/b98e55d/ : SUCCESS in 1h 04m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/3/check/openstack-tox-functional-py35/bbd86c3/ : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/3/check/openstack-tox-functional-py36/7c29931/ : SUCCESS in 6m 39s\n- cinder-rally-task http://logs.openstack.org/65/621465/3/check/cinder-rally-task/895c21c/ : SUCCESS in 42m 29s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/3/check/openstack-tox-pylint/1a5382b/ : FAILURE in 4m 20s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/3/check/devstack-plugin-ceph-tempest/1f61379/ : SUCCESS in 2h 03m 21s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/3/check/openstack-tox-functional/b3d5c8d/ : SUCCESS in 5m 06s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/3/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/30ae64b/ : SUCCESS in 1h 09m 29s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/3/check/legacy-tempest-dsvm-lvm-multibackend/0286f9e/ : SUCCESS in 1h 09m 13s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/3/check/legacy-tempest-dsvm-full-drbd-devstack/9e7f059/ : SUCCESS in 1h 29m 05s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/3/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/b2367c0/ : SUCCESS in 1h 16m 38s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/3/check/tempest-slow/3f31dcb/ : SUCCESS in 1h 53m 20s","accounts_in_message":[],"_revision_number":3},{"id":"473c103b110902147d8578d4f2be5f39c4910e57","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-04 09:20:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/3/check/dsvm-tempest-cisco-zonemanager-job/4eda230/ : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":3},{"id":"ef362ce5c971387b1c404d1c76bc107cdfb5169d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-04 09:24:34.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/3/EMC_VxFlexOS/550 : FAILURE in 7m 13s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"207e1969d1f89b358fb9e8a854c6d8d28b7deb68","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-04 09:37:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/3/check-cinder/Cinder-ISER-ISCSI/285098f : SUCCESS in 1h 05m 26s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/3/check-cinder/Cinder-ISER-LIO/c759461 : SUCCESS in 1h 00m 45s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"0ae4c89a5b9f093c6eac1b44a4fcf540ad1006a2","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-04 09:43:43.000000000","message":"Patch Set 3:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-3/nedge_iscsi : FAILURE in 24m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c739865da1daed326ed538c125982935bbd57a56","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-04 10:10:09.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/3/check/synology-dsm-cinder-iscsi/499efab : FAILURE in 40m 48s","accounts_in_message":[],"_revision_number":3},{"id":"957e96c6c8325bf8a95581f274e68e8a219240fb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-04 10:38:17.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"3ef21660f71eb8134500a66e13cffc785737f939","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-04 11:33:35.000000000","message":"Patch Set 3:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-3/ns5_iscsi : SUCCESS in 37m 55s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-3/ns5_nfs : SUCCESS in 41m 15s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-3/ns_nfs : SUCCESS in 64m 27s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-3/ns_iscsi : SUCCESS in 56m 6s","accounts_in_message":[],"_revision_number":3},{"id":"38427508c700b139f2b4bccbf03422c450b39144","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-04 12:24:12.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/3/check/dsvm-veritas-access-cinder-driver/d6807c6 : SUCCESS in 1h 32m 47s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/3/check/dsvm-veritas-access-cinder-iscsi-driver/e54c10d : FAILURE in 1h 54m 07s","accounts_in_message":[],"_revision_number":3},{"id":"2a36ba8ead1bd073b8f66ce4ac958b081e2877ba","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-04 14:20:26.000000000","message":"Patch Set 3:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/3/HPMSA-FC : SUCCESS in 35:24.46\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/3/HPMSA-iSCSI : SUCCESS in 35:58.13","accounts_in_message":[],"_revision_number":3},{"id":"a915b71a59e746f2b858872cbc20cb4898be0f17","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-04 14:26:18.000000000","message":"Patch Set 3:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/3/Lenovo-FC : SUCCESS in 35:48.30\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/3/Lenovo-iSCSI : SUCCESS in 34:59.70","accounts_in_message":[],"_revision_number":3},{"id":"11340c23641971b5e21bfd9989d0619c1410c2ca","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-04 15:25:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/3/check/kaminario-dsvm-tempest-full-iscsi/32d60f3 : SUCCESS in 1h 48m 31s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/3/check/kaminario-dsvm-tempest-full-FC/9fea73c : SUCCESS in 1h 44m 22s","accounts_in_message":[],"_revision_number":3},{"id":"e64657d65bcb6d47bb8e304771998ee09bde7217","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-04 15:52:09.000000000","message":"Patch Set 3:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e57df78 : SUCCESS in 2h 30m 55s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/3/check/3par-fc-driver-master-client-pip-ssa02-dsvm/b533946 : FAILURE in 2h 32m 24s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/3/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/2d4db21 : SUCCESS in 2h 25m 46s","accounts_in_message":[],"_revision_number":3},{"id":"27007698bc9e74fae2b75b9d821125260cd57e66","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-04 16:02:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0e8332776aa72be164461f32d561a73719f7e96b","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-04 16:12:57.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nHi Rajat, just verified that the uuid needs to be converted to uppercase before comparing.","accounts_in_message":[],"_revision_number":4},{"id":"634f0f24aa3a7eb69cb17a137ffdaa34f06edbac","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-04 16:18:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/4/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/ca4a216 : FAILURE in 15m 36s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/4/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/29a7a57 : FAILURE in 15m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"8eaa1b3a99fa85e46cb1fe5a8358556a79eff829","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-04 16:19:23.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/641 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/474 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"f967b1bf8a9813d97cf4b8062c211bfe7b78ee05","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-04 17:08:09.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/4/check/tempest-dsvm-full-xenial/2339e97/ : SUCCESS in 1h 04m 57s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/4/check/devstack-plugin-ceph-dsvm-xenial/2c41819/ : SUCCESS in 1h 03m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aa8b77b1e07930fc6b5d3ad2fc75f67adfdf9aea","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-04 17:18:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/4/iscsi/ : SUCCESS in 1h 14m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/4/smb/ : SUCCESS in 1h 11m 59s","accounts_in_message":[],"_revision_number":4},{"id":"c2f96a84c3a02c773c1a183abe22c505f3752723","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-04 17:22:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/4/check/dsvm-tempest-kvm/abf3c65 : SUCCESS in 1h 17m 40s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/4/check/dsvm-tempest-exe/6fc82dc : FAILURE in 5m 02s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"d5e6d1488a28062b5ef563611adeb1ab86f4e131","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-04 17:43:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/4/check/fujitsu-eternus-dx-iscsi/5fbe7fe : FAILURE in 1h 39m 27s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/4/check/fujitsu-eternus-dx-fc/c73763a : SUCCESS in 1h 38m 31s","accounts_in_message":[],"_revision_number":4},{"id":"8c9c9232fcc7f32fadbf183a9e120093f3c1268d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-04 17:52:29.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"d8541ffbeb2b16e808a25a9bfb89b4824204119e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-04 18:35:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/4/check/openstack-tox-lower-constraints/7172e51/ : SUCCESS in 11m 09s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/4/check/openstack-tox-pep8/a21835e/ : SUCCESS in 8m 38s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/4/check/openstack-tox-py27/f6ec96b/ : SUCCESS in 9m 32s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/4/check/openstack-tox-py35/ee25e0f/ : SUCCESS in 9m 15s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/4/check/openstack-tox-py36/fc8ffd1/ : SUCCESS in 10m 02s\n- openstack-tox-docs http://logs.openstack.org/65/621465/4/check/openstack-tox-docs/7b26d40/html/ : SUCCESS in 6m 32s\n- tempest-full http://logs.openstack.org/65/621465/4/check/tempest-full/fc48cf4/ : SUCCESS in 1h 37m 56s\n- neutron-grenade http://logs.openstack.org/65/621465/4/check/neutron-grenade/f8b71c1/ : SUCCESS in 55m 17s\n- grenade-py3 http://logs.openstack.org/65/621465/4/check/grenade-py3/c9ea402/ : SUCCESS in 1h 12m 02s\n- tempest-full-py3 http://logs.openstack.org/65/621465/4/check/tempest-full-py3/76f0802/ : SUCCESS in 1h 26m 45s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/4/check/cinder-tempest-dsvm-lvm-lio-barbican/3a92765/ : SUCCESS in 1h 47m 35s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/4/check/cinder-tox-bandit-baseline/4ff174b/ : SUCCESS in 8m 02s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/4/check/nova-multiattach/ec224c9/ : SUCCESS in 1h 02m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/4/check/openstack-tox-functional-py35/3114b46/ : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/4/check/openstack-tox-functional-py36/cd97164/ : SUCCESS in 5m 04s\n- cinder-rally-task http://logs.openstack.org/65/621465/4/check/cinder-rally-task/18c2aaa/ : SUCCESS in 34m 55s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/4/check/openstack-tox-pylint/8329c21/ : FAILURE in 4m 28s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/4/check/devstack-plugin-ceph-tempest/1095c3c/ : SUCCESS in 1h 59m 45s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/4/check/openstack-tox-functional/9432a8a/ : SUCCESS in 5m 49s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/4/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/d05b0c4/ : SUCCESS in 1h 02m 29s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/4/check/legacy-tempest-dsvm-lvm-multibackend/7c544c3/ : SUCCESS in 1h 28m 45s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/4/check/legacy-tempest-dsvm-full-drbd-devstack/f44bc5f/ : SUCCESS in 1h 18m 29s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/4/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2800966/ : SUCCESS in 1h 42m 22s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/4/check/tempest-slow/e863b91/ : SUCCESS in 2h 00m 55s","accounts_in_message":[],"_revision_number":4},{"id":"9a987e408b2caaf828703ce1e21478fe4104b7ba","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-04 18:37:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://10.63.152.76/logs/65/621465/4/upstream-solidfire-check/cinder-solidfire-iSCSI/a19182d/ : NOT_REGISTERED\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"8b388af0f81424fcd0854a12da279e208d150c2e","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-04 19:02:52.000000000","message":"Patch Set 4:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-4/nedge_iscsi : FAILURE in 24m 4s","accounts_in_message":[],"_revision_number":4},{"id":"d7d88fabac95add6034638dae437561da54c5c55","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-04 19:05:00.000000000","message":"Patch Set 4:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-4-iscsi : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"00486dfb7de1767a805f94c39c9e144e3269cbc4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-04 19:22:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://10.63.152.76/logs/65/621465/4/upstream-check/cinder-cDOT-iSCSI/b6f409a/ : SUCCESS in 26m 19s\n- cinder-cDOT-NFS http://10.63.152.76/logs/65/621465/4/upstream-check/cinder-cDOT-NFS/672c255/ : SUCCESS in 26m 47s\n- cinder-cDOT-FCP http://10.63.152.76/logs/65/621465/4/upstream-check/cinder-cDOT-FCP/799f39e/ : SUCCESS in 45m 34s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"adb9f52888448eabee2be598d0eb8255c8ca626b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-04 19:27:59.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"4f7a0501f4b99bc4ea6c0dce282672d314febff8","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-12-04 19:43:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://10.63.152.76/logs/65/621465/4/upstream-eseries-check/cinder-eseries-iSCSI/e81158c/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://10.63.152.76/logs/65/621465/4/upstream-eseries-check/cinder-eseries-FCP/6a061d3/ : SUCCESS in 1h 06m 15s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":4},{"id":"dbee9b40b9a699d16252c80a650d12e91f60a84c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-04 20:13:16.000000000","message":"Patch Set 4:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-4/ns5_iscsi : SUCCESS in 37m 49s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-4/ns5_nfs : SUCCESS in 39m 35s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-4/ns_nfs : SUCCESS in 65m 34s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-4/ns_iscsi : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":4},{"id":"e12224ec411c884437806131880742d746b5fd2c","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-05 00:35:32.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/4/check/dsvm-veritas-access-cinder-driver/fd748d6 : FAILURE in 1h 31m 04s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/4/check/dsvm-veritas-access-cinder-iscsi-driver/97f148f : FAILURE in 2h 51m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d5af6eb0bdb9b9c72f27ed3e0e175bbbd7c1a2ea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-05 00:49:32.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/4/EMC_VxFlexOS/558 : SUCCESS in 1h 47m 09s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"058d5372b93caf447a600508e970cca24cf28772","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-05 00:59:50.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/4/check-not-vote/ext-cinder-zuul/061924f : SUCCESS in 40m 18s","accounts_in_message":[],"_revision_number":4},{"id":"035de363cfc6427c041e698522ae3cbf20a0f7e6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-05 04:55:04.000000000","message":"Patch Set 4:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/32aff4c : FAILURE in 43m 04s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/4/check/3par-fc-driver-master-client-pip-ssa02-dsvm/d9c7829 : FAILURE in 37m 52s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/4/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/f8a853a : FAILURE in 33m 15s","accounts_in_message":[],"_revision_number":4},{"id":"178db5964f398bebae247db20b59bd59586fa0aa","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-05 08:58:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"50199c5fb4737fdae3d563658a77012e9ac7f29c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-05 09:48:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/5/check/fujitsu-eternus-dx-iscsi/9a4f437 : FAILURE in 43m 40s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/5/check/fujitsu-eternus-dx-fc/f326c89 : FAILURE in 1m 15s","accounts_in_message":[],"_revision_number":5},{"id":"d8e0e728cf8068dbbde8a7521644d220e1ba69ba","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-05 09:59:32.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/661 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/494 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"cfc14f04a8863c4e337fd63e7d62f0b028b2b830","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-05 10:06:26.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/5/check/tempest-dsvm-full-xenial/2066973/ : SUCCESS in 1h 02m 25s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/5/check/devstack-plugin-ceph-dsvm-xenial/dbc4d59/ : SUCCESS in 1h 07m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"64da4f8765dde1897766606c19da717bd5e095c6","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-05 10:36:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/5/iscsi/ : SUCCESS in 50m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/5/smb/ : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":5},{"id":"f5c436a3e8eef15e7deb604524c3b227e08d64d8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-05 10:50:03.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/5/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/96787de : FAILURE in 14m 58s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/5/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/d4e41cb : FAILURE in 14m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"8758f652dcb48b3fd0623e70b484063077bcf0f7","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-05 10:54:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/5/check/dsvm-tempest-kvm/7e3b3ff : SUCCESS in 1h 03m 22s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/5/check/dsvm-tempest-exe/2a1bd95 : SUCCESS in 47m 29s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"43522bfc8fd5d7fc8c803f96eba44a354c980ffe","author":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"date":"2018-12-05 12:03:59.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3ae3ce59978d270d6f36d13a713a1657b2cd9f70","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-05 12:05:29.000000000","message":"Patch Set 4:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/4/HPMSA-FC : SUCCESS in 35:38.09\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/4/HPMSA-iSCSI : SUCCESS in 56:59.98","accounts_in_message":[],"_revision_number":4},{"id":"6a4ddc82506df2ce6a57ef394e4d66c3bc7262a4","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2018-12-05 13:16:55.000000000","message":"Patch Set 5: Code-Review-1\n\nIn your driver you are mixing volume driver functionality with target driver functionality (cinder/volume/targets/). I think it would be a better idea to create separate target driver here. Target driver should also inherit from nvmeof.NVMeOF class. This way it will simplify this code.","accounts_in_message":[],"_revision_number":5},{"id":"9a67043efd6aa7142ac2e1e2af0c2ec3440e3eb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-05 14:10:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/5/check/openstack-tox-lower-constraints/e536277/ : SUCCESS in 11m 11s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/5/check/openstack-tox-pep8/1a92fca/ : FAILURE in 8m 03s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/5/check/openstack-tox-py27/9dbe534/ : SUCCESS in 9m 55s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/5/check/openstack-tox-py35/0427567/ : SUCCESS in 9m 47s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/5/check/openstack-tox-py36/6b8f4b4/ : SUCCESS in 10m 36s\n- openstack-tox-docs http://logs.openstack.org/65/621465/5/check/openstack-tox-docs/d6d6860/html/ : SUCCESS in 7m 04s\n- tempest-full http://logs.openstack.org/65/621465/5/check/tempest-full/618476f/ : SUCCESS in 1h 25m 38s\n- neutron-grenade http://logs.openstack.org/65/621465/5/check/neutron-grenade/2f13952/ : SUCCESS in 50m 45s\n- grenade-py3 http://logs.openstack.org/65/621465/5/check/grenade-py3/025edab/ : SUCCESS in 53m 46s\n- tempest-full-py3 http://logs.openstack.org/65/621465/5/check/tempest-full-py3/7712758/ : SUCCESS in 1h 20m 23s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/5/check/cinder-tempest-dsvm-lvm-lio-barbican/69b9350/ : SUCCESS in 1h 35m 43s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/5/check/cinder-tox-bandit-baseline/7d8a8f6/ : SUCCESS in 8m 36s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/5/check/nova-multiattach/8b4b8d6/ : SUCCESS in 58m 26s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/5/check/openstack-tox-functional-py35/c5c67f5/ : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/5/check/openstack-tox-functional-py36/a59aa18/ : SUCCESS in 7m 14s\n- cinder-rally-task http://logs.openstack.org/65/621465/5/check/cinder-rally-task/c4bdbfe/ : SUCCESS in 38m 27s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/5/check/openstack-tox-pylint/26e06ae/ : FAILURE in 6m 35s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/5/check/devstack-plugin-ceph-tempest/c1b8e81/ : SUCCESS in 1h 37m 05s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/5/check/openstack-tox-functional/9155f1c/ : SUCCESS in 6m 28s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/5/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/cb9034f/ : SUCCESS in 1h 00m 37s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/5/check/legacy-tempest-dsvm-lvm-multibackend/ae0b7fa/ : SUCCESS in 1h 18m 36s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/5/check/legacy-tempest-dsvm-full-drbd-devstack/f8a91eb/ : SUCCESS in 1h 21m 37s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/5/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/066e229/ : SUCCESS in 1h 25m 57s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/5/check/tempest-slow/0f0f16d/ : SUCCESS in 1h 48m 15s","accounts_in_message":[],"_revision_number":5},{"id":"5b9cd98b002b57bda13135bf029f86c1afde30b7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-05 15:14:26.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/5/check-cinder/Cinder-ISER-ISCSI/8aa602a : SUCCESS in 55m 56s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/5/check-cinder/Cinder-ISER-LIO/9c87d06 : SUCCESS in 1h 01m 53s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"6b94b9cb398d70946b3c56112c02973b08aad691","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-05 15:26:29.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/5/check-not-vote/ext-cinder-zuul/c9bb433 : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":5},{"id":"9d41b3e6f8e970b4b7918f5dd804b10c6408c7d1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2018-12-05 15:28:30.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3757a3253c7de87f2c66bf9cf225b31b7d9310db","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-05 20:49:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/4/check/NEC-ISCSI/76b49eb : SUCCESS in 1h 26m 18s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/4/check/NEC-FC/2c80a2c : SUCCESS in 1h 26m 28s","accounts_in_message":[],"_revision_number":4},{"id":"7204490831fa681a3d68235a3102df22ed59fd3c","author":{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},"date":"2018-12-06 01:19:52.000000000","message":"Patch Set 3:\n\n@Maciej Szwed, the connection-related interfaces are also defined in BaseVD class. I think when a target is sharable between different storage devices it is better to split as a standalone target. This can refer to existed drivers. Currently, RSD target is only used by RSD, so let\u0027s keep the current method, and update it later if needed.","accounts_in_message":[],"_revision_number":3},{"id":"c674018061ac8b9b0608b9be3a81c18251316d54","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-06 02:00:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"49ef300abcb306176f79be07c20397569a553fed","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-06 02:23:37.000000000","message":"Patch Set 5:\n\n(4 comments)\n\nThanks David and Eric, all addressed.\n\n@Maciej: I agree with Lisa\u0027s comments. If there is a need for another driver to reuse RSD target, we can then evaluate both to figure out how to split the code. Currently we are focusing on the driver enabling with mandatory features.","accounts_in_message":[],"_revision_number":5},{"id":"bc42216b6905d4d18bac2efe4313567f145dc4a3","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-06 02:33:54.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/668 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/501 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"741f4b5c5ff7fd1a898f2e7cace8b071fdfeccdc","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-06 02:56:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/6/check/fujitsu-eternus-dx-iscsi/5c45b38 : FAILURE in 46m 12s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/6/check/fujitsu-eternus-dx-fc/fbe1437 : FAILURE in 45m 18s","accounts_in_message":[],"_revision_number":6},{"id":"3fdf11537194672c728241af8e1df14f710684a4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-06 03:08:54.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/6/check/tempest-dsvm-full-xenial/0e5e782/ : SUCCESS in 1h 07m 50s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/6/check/devstack-plugin-ceph-dsvm-xenial/5b50f5d/ : SUCCESS in 1h 07m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eb9886cc2daefbbd75dd9b62446b1ad28f43a840","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-06 03:36:42.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/6/check/dsvm-tempest-kvm/5634c84 : SUCCESS in 1h 08m 52s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/6/check/dsvm-tempest-exe/5847f30 : FAILURE in 38m 17s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"50d483d145a6fcf25996b54ef3d06d0649265e27","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-06 03:55:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/6/iscsi/ : SUCCESS in 51m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/6/smb/ : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":6},{"id":"821651e46f3b2b0ac56bb09b13ddcbd982d65010","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-06 03:56:21.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/6/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f3282fe : FAILURE in 17m 26s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/6/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/2a6b729 : FAILURE in 15m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"333fd17d2486179b34275c344e8d99e263d45458","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-06 05:12:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/6/check/openstack-tox-lower-constraints/80e4e06/ : SUCCESS in 14m 00s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/6/check/openstack-tox-pep8/97b9842/ : FAILURE in 8m 13s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/6/check/openstack-tox-py27/6d4fb57/ : SUCCESS in 9m 38s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/6/check/openstack-tox-py35/d882697/ : SUCCESS in 8m 52s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/6/check/openstack-tox-py36/fd5efd7/ : SUCCESS in 8m 15s\n- openstack-tox-docs http://logs.openstack.org/65/621465/6/check/openstack-tox-docs/094ecda/html/ : SUCCESS in 6m 55s\n- tempest-full http://logs.openstack.org/65/621465/6/check/tempest-full/a37a91b/ : SUCCESS in 1h 33m 52s\n- neutron-grenade http://logs.openstack.org/65/621465/6/check/neutron-grenade/90ad5d0/ : SUCCESS in 52m 09s\n- grenade-py3 http://logs.openstack.org/65/621465/6/check/grenade-py3/b9e987f/ : SUCCESS in 54m 52s\n- tempest-full-py3 http://logs.openstack.org/65/621465/6/check/tempest-full-py3/5b37c1b/ : SUCCESS in 1h 31m 10s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/6/check/cinder-tempest-dsvm-lvm-lio-barbican/886314c/ : SUCCESS in 1h 34m 20s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/6/check/cinder-tox-bandit-baseline/3940114/ : SUCCESS in 7m 31s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/6/check/nova-multiattach/196687d/ : SUCCESS in 54m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/6/check/openstack-tox-functional-py35/46fc939/ : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/6/check/openstack-tox-functional-py36/eca2d8d/ : SUCCESS in 5m 38s\n- cinder-rally-task http://logs.openstack.org/65/621465/6/check/cinder-rally-task/4d5e0b8/ : SUCCESS in 40m 17s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/6/check/openstack-tox-pylint/9afb237/ : FAILURE in 4m 43s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/6/check/devstack-plugin-ceph-tempest/b7a6f12/ : SUCCESS in 1h 49m 24s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/6/check/openstack-tox-functional/896706b/ : SUCCESS in 5m 45s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/6/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/cb879bd/ : SUCCESS in 1h 06m 41s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/6/check/legacy-tempest-dsvm-lvm-multibackend/f34410b/ : SUCCESS in 1h 06m 39s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/6/check/legacy-tempest-dsvm-full-drbd-devstack/7aca108/ : SUCCESS in 1h 20m 48s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/6/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/78bc76c/ : SUCCESS in 1h 14m 28s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/6/check/tempest-slow/5ce0229/ : SUCCESS in 1h 38m 49s","accounts_in_message":[],"_revision_number":6},{"id":"7de4aeff3e723e2c89dcb6ac99a425a8536ceb96","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-06 05:32:32.000000000","message":"Patch Set 4:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/4/Lenovo-FC : FAILURE in 5:37:22\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/4/Lenovo-iSCSI : FAILURE in 5:27:55","accounts_in_message":[],"_revision_number":4},{"id":"a1b2b234827bd90e29105f77f016d24d1257f4b7","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-06 06:01:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3c3625d0287df644649ae97a64ff816ab57d6ac4","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-06 06:18:58.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/672 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/505 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"8e629ba5369e25cfb91497cbc9c742c9f6ce98d5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-06 06:22:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/7/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/0f37b0b : FAILURE in 15m 10s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/7/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/9183cff : FAILURE in 15m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"6837277b5c840749b4513819251bcc97441f36c4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-06 06:47:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/7/check/fujitsu-eternus-dx-iscsi/5b7c9e5 : FAILURE in 39m 49s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/7/check/fujitsu-eternus-dx-fc/94e1691 : FAILURE in 40m 55s","accounts_in_message":[],"_revision_number":7},{"id":"74b4b70d71f3fd4bc7b946c6a4ec9858ca911120","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-06 06:50:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/7/iscsi/ : SUCCESS in 48m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/7/smb/ : SUCCESS in 46m 01s","accounts_in_message":[],"_revision_number":7},{"id":"46650336a938a429df2dcf72423f138e786a9568","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-06 07:07:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/7/check-cinder/Cinder-ISER-ISCSI/9fb8ff5 : SUCCESS in 57m 00s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/7/check-cinder/Cinder-ISER-LIO/2f50245 : SUCCESS in 1h 05m 31s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"593e7b7cd85618b4327e7153af1dfd3fc06cc68a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-06 07:09:19.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/7/check/tempest-dsvm-full-xenial/1930a9d/ : SUCCESS in 1h 01m 21s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/7/check/devstack-plugin-ceph-dsvm-xenial/7141e77/ : SUCCESS in 1h 07m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"89a15f39ffa08121f4a04467829bb7d593011f6c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-06 07:29:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/7/check/dsvm-tempest-kvm/744f156 : SUCCESS in 1h 11m 15s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/7/check/dsvm-tempest-exe/4082823 : FAILURE in 24m 07s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"8bcd58f4582611c2c74956e7872a1cf01a27fe34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-06 08:29:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/7/check/openstack-tox-lower-constraints/7450310/ : SUCCESS in 10m 22s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/7/check/openstack-tox-pep8/828fd20/ : SUCCESS in 9m 08s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/7/check/openstack-tox-py27/ea59594/ : SUCCESS in 9m 38s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/7/check/openstack-tox-py35/d0faef5/ : SUCCESS in 9m 55s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/7/check/openstack-tox-py36/6c0c319/ : SUCCESS in 8m 50s\n- openstack-tox-docs http://logs.openstack.org/65/621465/7/check/openstack-tox-docs/eb17db3/html/ : SUCCESS in 8m 03s\n- tempest-full http://logs.openstack.org/65/621465/7/check/tempest-full/00abe09/ : SUCCESS in 1h 21m 00s\n- neutron-grenade http://logs.openstack.org/65/621465/7/check/neutron-grenade/25dda82/ : SUCCESS in 1h 03m 13s\n- grenade-py3 http://logs.openstack.org/65/621465/7/check/grenade-py3/5d7ee4f/ : SUCCESS in 59m 54s\n- tempest-full-py3 http://logs.openstack.org/65/621465/7/check/tempest-full-py3/b0a7d3e/ : SUCCESS in 1h 19m 21s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/7/check/cinder-tempest-dsvm-lvm-lio-barbican/d03729c/ : SUCCESS in 1h 48m 58s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/7/check/cinder-tox-bandit-baseline/f440bab/ : SUCCESS in 8m 03s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/7/check/nova-multiattach/d9f07ea/ : SUCCESS in 1h 05m 52s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/7/check/openstack-tox-functional-py35/00fe49d/ : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/7/check/openstack-tox-functional-py36/bdf2b19/ : SUCCESS in 5m 25s\n- cinder-rally-task http://logs.openstack.org/65/621465/7/check/cinder-rally-task/1dc2653/ : SUCCESS in 39m 02s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/7/check/openstack-tox-pylint/8de7d95/ : FAILURE in 4m 49s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/7/check/devstack-plugin-ceph-tempest/b917eca/ : SUCCESS in 1h 51m 30s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/7/check/openstack-tox-functional/41a7721/ : SUCCESS in 5m 27s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/7/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/e1ab1c3/ : SUCCESS in 1h 10m 49s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/7/check/legacy-tempest-dsvm-lvm-multibackend/fca82c8/ : SUCCESS in 1h 27m 16s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/7/check/legacy-tempest-dsvm-full-drbd-devstack/572cf25/ : SUCCESS in 1h 17m 28s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/7/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/0e6bbf1/ : SUCCESS in 1h 31m 36s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/7/check/tempest-slow/d5adfd5/ : SUCCESS in 1h 52m 17s","accounts_in_message":[],"_revision_number":7},{"id":"e80fae0035277393ba67daa3ca5fc6b202ff4f21","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-06 09:50:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/7/check/dsvm-tempest-cisco-zonemanager-job/47bf7da/ : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":7},{"id":"316face53b2344233f9b3a7e102fd63ed28b645b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-06 09:56:16.000000000","message":"Patch Set 7:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"d820c8c0c108c4ba1bbba4c778c2aa69cedafc1d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-06 09:59:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://10.63.152.76/logs/65/621465/7/upstream-check/cinder-cDOT-iSCSI/463b1eb/ : SUCCESS in 26m 21s\n- cinder-cDOT-NFS http://10.63.152.76/logs/65/621465/7/upstream-check/cinder-cDOT-NFS/731af03/ : SUCCESS in 1h 29m 55s\n- cinder-cDOT-FCP http://10.63.152.76/logs/65/621465/7/upstream-check/cinder-cDOT-FCP/f8b056d/ : FAILURE in 52m 12s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"4953ded0b324a1fe257dbd27aaf2fb2c5e8552de","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-06 10:05:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://10.63.152.76/logs/65/621465/7/upstream-solidfire-check/cinder-solidfire-iSCSI/259bdc5/ : SUCCESS in 1h 34m 39s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"17b350a4986781cf01d18b874c415683ff73c400","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-12-06 11:03:27.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://10.63.152.76/logs/65/621465/7/upstream-eseries-check/cinder-eseries-iSCSI/636814c/ : FAILURE in 2h 33m 20s (non-voting)\n- cinder-eseries-FCP http://10.63.152.76/logs/65/621465/7/upstream-eseries-check/cinder-eseries-FCP/bced7e5/ : POST_FAILURE in 2m 14s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":7},{"id":"fd26343526570e27b58b414a33923e22f3841027","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-06 12:36:00.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/7/check/EMC_Unity_FC/65e86c3/EMC_Unity_FC/400 : SUCCESS in 1h 15m 42s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/7/check/EMC_Unity_iSCSI/225f510/EMC_Unity_iSCSI/386 : FAILURE in 1h 53m 59s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"ca6c7e49de6c443139a6f0eebf160487eb1b222f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-06 12:43:47.000000000","message":"Patch Set 7:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c21f856 : FAILURE in 1h 43m 44s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/7/check/3par-fc-driver-master-client-pip-ssa02-dsvm/d36467b : FAILURE in 44m 17s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/7/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/8b2d010 : SUCCESS in 1h 53m 23s","accounts_in_message":[],"_revision_number":7},{"id":"87a7dd36501a49e462c79c04c38787277fc473de","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-06 13:16:45.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/7/check/synology-dsm-cinder-iscsi/c382b72 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":7},{"id":"dc2a77ac34a57d9c836da30b3d316f1204f2fcd3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-06 14:54:00.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/7/EMC_VxFlexOS/573 : SUCCESS in 1h 53m 16s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"86a9fc9c1c98f634b57a607e5276657e4a960706","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-06 16:31:18.000000000","message":"Patch Set 4:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4070c45 : FAILURE in 2h 45m 24s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/4/check/3par-fc-driver-master-client-pip-ssa02-dsvm/8d92044 : SUCCESS in 1h 47m 23s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/4/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/02a5bf6 : SUCCESS in 2h 04m 24s","accounts_in_message":[],"_revision_number":4},{"id":"130ab819aacdf281e3e4941da0822392d6fc83b4","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-06 21:24:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/7/check/kaminario-dsvm-tempest-full-iscsi/9867325 : SUCCESS in 2h 04m 00s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/7/check/kaminario-dsvm-tempest-full-FC/8163cb8 : SUCCESS in 1h 47m 15s","accounts_in_message":[],"_revision_number":7},{"id":"48cea142743c8314df178cc1ba7d3f92e2c690ff","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-06 23:40:46.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/7/check/dsvm-veritas-access-cinder-driver/4ff6517 : FAILURE in 1h 37m 20s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/7/check/dsvm-veritas-access-cinder-iscsi-driver/aa06c9f : FAILURE in 2h 46m 44s","accounts_in_message":[],"_revision_number":7},{"id":"a6c982af15bd2722e31a1110cf8eced91f355286","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-07 03:11:27.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fb46687d59c83098cabddd1e81d9b415af3a2b11","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-07 03:29:23.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/689 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/522 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"b7982801fb8d74e2942c7fc621e2b1b991e7b23c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-07 03:40:23.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/8/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/31a1d36 : FAILURE in 14m 51s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/8/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/dcb0fcf : FAILURE in 15m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"93d0c0230130ef8241c9913c4d1a7c5b780152ac","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-07 04:04:14.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/8/iscsi/ : SUCCESS in 52m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/8/smb/ : FAILURE in 51m 23s","accounts_in_message":[],"_revision_number":8},{"id":"8d51a6c9b37e05e864c04699ddb5d26567c71a5d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-07 04:19:55.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/8/check/tempest-dsvm-full-xenial/cdfa7f4/ : SUCCESS in 1h 08m 09s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/8/check/devstack-plugin-ceph-dsvm-xenial/dbdd78f/ : SUCCESS in 1h 04m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"199b52d532b55cc471c83b95d7971685669f9150","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-07 04:50:52.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/8/check/dsvm-tempest-cisco-zonemanager-job/2e1d10c/ : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":8},{"id":"dff2b818389576ce835af2c79c3e17411543b0ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 05:33:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/8/check/openstack-tox-lower-constraints/a51a0c9/ : SUCCESS in 10m 32s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/8/check/openstack-tox-pep8/10d9222/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/8/check/openstack-tox-py27/a9703b0/ : SUCCESS in 8m 36s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/8/check/openstack-tox-py35/46e3839/ : SUCCESS in 8m 54s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/8/check/openstack-tox-py36/130a65f/ : SUCCESS in 8m 12s\n- openstack-tox-docs http://logs.openstack.org/65/621465/8/check/openstack-tox-docs/18b5670/html/ : SUCCESS in 6m 22s\n- tempest-full http://logs.openstack.org/65/621465/8/check/tempest-full/bd45135/ : FAILURE in 1h 15m 23s\n- neutron-grenade http://logs.openstack.org/65/621465/8/check/neutron-grenade/c89c0ec/ : SUCCESS in 1h 26m 25s\n- grenade-py3 http://logs.openstack.org/65/621465/8/check/grenade-py3/3aa7a70/ : SUCCESS in 1h 06m 38s\n- tempest-full-py3 http://logs.openstack.org/65/621465/8/check/tempest-full-py3/7b7a78a/ : SUCCESS in 1h 36m 37s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/8/check/cinder-tempest-dsvm-lvm-lio-barbican/3387d07/ : SUCCESS in 2h 04m 10s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/8/check/cinder-tox-bandit-baseline/3d0c458/ : SUCCESS in 7m 07s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/8/check/nova-multiattach/66569bf/ : SUCCESS in 50m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/8/check/openstack-tox-functional-py35/a41c596/ : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/8/check/openstack-tox-functional-py36/626910e/ : SUCCESS in 4m 58s\n- cinder-rally-task http://logs.openstack.org/65/621465/8/check/cinder-rally-task/eac66c0/ : SUCCESS in 41m 24s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/8/check/openstack-tox-pylint/059a592/ : FAILURE in 4m 13s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/8/check/devstack-plugin-ceph-tempest/5a4a840/ : SUCCESS in 1h 42m 15s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/8/check/openstack-tox-functional/cb3a7a0/ : SUCCESS in 5m 02s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/8/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/8800c45/ : SUCCESS in 57m 25s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/8/check/legacy-tempest-dsvm-lvm-multibackend/a21417c/ : SUCCESS in 1h 14m 47s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/8/check/legacy-tempest-dsvm-full-drbd-devstack/73da23c/ : SUCCESS in 1h 20m 55s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/8/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/5d9ff0e/ : SUCCESS in 1h 36m 09s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/8/check/tempest-slow/f9c7b26/ : SUCCESS in 1h 48m 24s","accounts_in_message":[],"_revision_number":8},{"id":"96e87ae4e7af5bcc248088ff78cca28fbd69ca21","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-07 05:33:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/8/check/dsvm-tempest-kvm/c3940f3 : SUCCESS in 1h 09m 00s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/8/check/dsvm-tempest-exe/0190591 : FAILURE in 43m 58s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"96111d2ba2ca20ddbb443b2e6206dffc90a92bde","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-12-07 05:33:41.000000000","message":"Patch Set 8:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- tempest-full: https://bugs.launchpad.net/bugs/1737634\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":8},{"id":"6aa9051475344b8b45e51fbfebc9c7daea40d635","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-07 05:34:22.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/8/check/fujitsu-eternus-dx-iscsi/f2fe0b9 : SUCCESS in 2h 19m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/8/check/fujitsu-eternus-dx-fc/fe30cca : FAILURE in 1h 06m 14s","accounts_in_message":[],"_revision_number":8},{"id":"1716a4b3447957e4c1537b87d129444087a97910","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-07 06:24:55.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"3874026b1c0d76cf6602e430ddfdb4bf965096fb","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-07 06:43:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"99c4b72a649897de8a30c841b19325f388be32bf","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-07 06:45:52.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ae096a2bc7abf6ebb989ce17e5d71aa598ba5d69","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-07 07:00:53.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/690 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/523 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"f8057fd525e141988bdc50b2d94d42bf17029a20","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-07 07:18:00.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/691 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/524 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"f1401b61e1ef50dc6c5564ec92414c16b9eb23dd","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-07 07:30:49.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/10/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/8eb4401 : FAILURE in 22m 29s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/10/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e919c94 : FAILURE in 15m 29s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"03e7417a78ae133941b4bbe0e927b7fad7c3ac91","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-07 07:41:37.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/10/check/dsvm-tempest-kvm/54cbac4 : SUCCESS in 55m 05s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/10/check/dsvm-tempest-exe/1403e66 : FAILURE in 40m 42s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"793598e972a0496042d780ab3869a99e7d3f6158","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-07 07:46:29.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/10/iscsi/ : SUCCESS in 1h 00m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/10/smb/ : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":10},{"id":"59a4595ee989c2068e1a34a04385449ab7c6edc9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-07 07:56:46.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/10/check/tempest-dsvm-full-xenial/225e032/ : SUCCESS in 1h 08m 49s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/10/check/devstack-plugin-ceph-dsvm-xenial/2ddf6a3/ : SUCCESS in 1h 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d0a03550135b4a7feb3795d9ee3bb0f30e195028","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-07 08:01:48.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"843a2d081be4d7351f4f631dd4c4782c159f99ed","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-07 08:07:03.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/10/check/dsvm-tempest-cisco-zonemanager-job/4e3474c/ : FAILURE in 31m 28s","accounts_in_message":[],"_revision_number":10},{"id":"0fb19aa07d2174d27f96d14c30dd921f71147fa8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-07 08:08:54.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"3a86925d227d25ee3397ead0378eb44fd4749199","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-07 08:18:38.000000000","message":"Patch Set 7:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/7/Lenovo-FC : SUCCESS in 36:20.91\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/7/Lenovo-iSCSI : SUCCESS in 36:36.36","accounts_in_message":[],"_revision_number":7},{"id":"1c3214c5915e05ee178351f3f859c362fecd8c4d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-07 08:41:23.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/10/check-cinder/Cinder-ISER-ISCSI/ef0036a : SUCCESS in 55m 33s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/10/check-cinder/Cinder-ISER-LIO/bbcb3ee : SUCCESS in 55m 56s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"810f9631c5b86e9e5aa80eedab47b659b7ffd7ef","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-07 09:24:49.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/10/check/fujitsu-eternus-dx-iscsi/99782e2 : FAILURE in 1h 10m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/10/check/fujitsu-eternus-dx-fc/61e6ab6 : SUCCESS in 2h 37m 12s","accounts_in_message":[],"_revision_number":10},{"id":"0be8e71bc452bf299e8603cae856c9a57ad29bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 09:26:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/10/check/openstack-tox-lower-constraints/108415b/ : SUCCESS in 12m 57s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/10/check/openstack-tox-pep8/9531e82/ : SUCCESS in 9m 04s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/10/check/openstack-tox-py27/75639ad/ : SUCCESS in 8m 56s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/10/check/openstack-tox-py35/b99e772/ : SUCCESS in 9m 51s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/10/check/openstack-tox-py36/9a3a108/ : SUCCESS in 8m 31s\n- openstack-tox-docs http://logs.openstack.org/65/621465/10/check/openstack-tox-docs/a445d78/html/ : SUCCESS in 7m 16s\n- tempest-full http://logs.openstack.org/65/621465/10/check/tempest-full/3da9a6c/ : SUCCESS in 1h 35m 17s\n- neutron-grenade http://logs.openstack.org/65/621465/10/check/neutron-grenade/f97b1c6/ : SUCCESS in 59m 31s\n- grenade-py3 http://logs.openstack.org/65/621465/10/check/grenade-py3/c79bdaf/ : SUCCESS in 1h 01m 56s\n- tempest-full-py3 http://logs.openstack.org/65/621465/10/check/tempest-full-py3/18a3559/ : SUCCESS in 1h 21m 17s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/10/check/build-openstack-releasenotes/108a67e/html/ : SUCCESS in 15m 02s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/10/check/cinder-tempest-dsvm-lvm-lio-barbican/41b2ed9/ : SUCCESS in 1h 50m 26s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/10/check/cinder-tox-bandit-baseline/ed88f45/ : SUCCESS in 7m 56s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/10/check/nova-multiattach/e6fa741/ : SUCCESS in 55m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/10/check/openstack-tox-functional-py35/d240825/ : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/10/check/openstack-tox-functional-py36/32c2fdc/ : SUCCESS in 6m 36s\n- cinder-rally-task http://logs.openstack.org/65/621465/10/check/cinder-rally-task/01f4340/ : SUCCESS in 38m 06s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/10/check/openstack-tox-pylint/c064c07/ : FAILURE in 4m 35s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/10/check/devstack-plugin-ceph-tempest/326cef4/ : SUCCESS in 1h 40m 20s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/10/check/openstack-tox-functional/c6351ca/ : SUCCESS in 5m 18s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/10/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/eae9f10/ : SUCCESS in 1h 08m 54s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/10/check/legacy-tempest-dsvm-lvm-multibackend/31b6588/ : SUCCESS in 1h 14m 20s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/10/check/legacy-tempest-dsvm-full-drbd-devstack/8d9f6fd/ : SUCCESS in 1h 08m 50s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/10/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/6bcf294/ : FAILURE in 2h 27m 25s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/10/check/tempest-slow/80ab2e6/ : SUCCESS in 1h 45m 09s","accounts_in_message":[],"_revision_number":10},{"id":"de053f27a9fc7438ac6fadcade268256b8b52a50","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-07 09:30:50.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/10/check/EMC_XtremIO_FC/103bc81/EMC_XtremIO_FC/944 : SUCCESS in 1h 29m 45s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/10/check/EMC_XtremIO_iSCSI/c406cca/EMC_XtremIO_iSCSI/965 : SUCCESS in 1h 22m 26s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":10},{"id":"a2e41ae7ab375fe10bf5a74cbd712629e1f40332","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-07 10:02:35.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://10.63.152.76/logs/65/621465/10/upstream-solidfire-check/cinder-solidfire-iSCSI/aef8197/ : SUCCESS in 34m 43s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"50fcdebb97d07f94b5fbefb8dfcd084e779b6afb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-07 10:10:34.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://10.63.152.76/logs/65/621465/10/upstream-check/cinder-cDOT-iSCSI/b524780/ : SUCCESS in 26m 42s\n- cinder-cDOT-NFS http://10.63.152.76/logs/65/621465/10/upstream-check/cinder-cDOT-NFS/da24bc8/ : SUCCESS in 26m 57s\n- cinder-cDOT-FCP http://10.63.152.76/logs/65/621465/10/upstream-check/cinder-cDOT-FCP/5c6ada8/ : SUCCESS in 43m 19s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"8f891d935af1d565808a21ed90d616a4473d6009","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-07 10:11:01.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/10/check/synology-dsm-cinder-iscsi/2b793f2 : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":10},{"id":"928ce44bb71984c7da879603193b9831caf66459","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-07 10:20:23.000000000","message":"Patch Set 10:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-10 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":10},{"id":"3d6d5b1f4c433d46e37fc7810a265726bcc4da3f","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-12-07 11:07:12.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://10.63.152.76/logs/65/621465/10/upstream-eseries-check/cinder-eseries-iSCSI/8b51083/ : FAILURE in 1h 36m 39s (non-voting)\n- cinder-eseries-FCP http://10.63.152.76/logs/65/621465/10/upstream-eseries-check/cinder-eseries-FCP/388374d/ : SUCCESS in 1h 33m 21s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":10},{"id":"1da89bf95f5dabf491ee9cc86b1227e634aa95d8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-07 13:17:24.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/10/EMC_VxFlexOS/585 : SUCCESS in 1h 50m 43s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":10},{"id":"a46610e7a930726854a0c1a47b9cc78dd439df90","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-07 15:17:32.000000000","message":"Patch Set 10:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/945f68b : FAILURE in 1h 35m 33s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/10/check/3par-fc-driver-master-client-pip-ssa02-dsvm/f5aa66b : FAILURE in 1h 30m 31s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/10/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/48c8cd9 : SUCCESS in 1h 57m 10s","accounts_in_message":[],"_revision_number":10},{"id":"9941ee3ee184683c5b4b5565d812e9cc1e417520","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-07 16:28:59.000000000","message":"Patch Set 7:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/7/HPMSA-FC : SUCCESS in 36:16.75\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/7/HPMSA-iSCSI : SUCCESS in 37:07.04","accounts_in_message":[],"_revision_number":7},{"id":"9a6761fe69cf26ff342ccc641cc17fcdb36036d4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-07 17:30:45.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/10/check/EMC_Unity_FC/1550a09/EMC_Unity_FC/418 : SUCCESS in 1h 12m 49s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/10/check/EMC_Unity_iSCSI/751f829/EMC_Unity_iSCSI/403 : SUCCESS in 1h 29m 43s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"4228c97286bce67ef41a918c91aaded764a597ce","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-07 19:14:51.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/10/check/dsvm-veritas-access-cinder-driver/326e3f1 : FAILURE in 1h 31m 21s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/10/check/dsvm-veritas-access-cinder-iscsi-driver/1b48d6f : FAILURE in 1h 57m 56s","accounts_in_message":[],"_revision_number":10},{"id":"7cbc5aa9b32205e3e2c62fb79b3ea54f19485aa5","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-08 01:18:12.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/10/check/kaminario-dsvm-tempest-full-iscsi/1504703 : SUCCESS in 1h 48m 40s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/10/check/kaminario-dsvm-tempest-full-FC/65013fb : SUCCESS in 1h 47m 55s","accounts_in_message":[],"_revision_number":10},{"id":"bf91d674f2e9d504ceccbdecbf8c1d1e656f9ffc","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-08 08:46:55.000000000","message":"Patch Set 10:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/10/Lenovo-FC : SUCCESS in 42:22.95\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/10/Lenovo-iSCSI : SUCCESS in 1:00:11","accounts_in_message":[],"_revision_number":10},{"id":"fa35aa7ec1a44c40746994c8da40c1bc18eb58a2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-08 11:02:42.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/10/check/dsvm-tempest-storpool/5aa189c : SUCCESS in 1h 37m 50s","accounts_in_message":[],"_revision_number":10},{"id":"2997c39d5811019659dd67614882af165e2ce279","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2018-12-08 15:18:21.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_FC http://publiclogs.emc.com/621465/10/EMC_VMAX_FC/635 : FAILURE in 7h 59m 52s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/621465/10/EMC_VMAX_iSCSI/607 : FAILURE in 7h 58m 05s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":10},{"id":"ccc897d0e557b7317f4a455c209ab6d030548de9","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-08 15:50:58.000000000","message":"Patch Set 10:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/10/HPMSA-FC : SUCCESS in 35:29.04\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/10/HPMSA-iSCSI : SUCCESS in 36:23.91","accounts_in_message":[],"_revision_number":10},{"id":"2453dc863e6fb3cd3151d455759a8d579024d8ce","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2018-12-08 18:34:17.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/621465/10/Dell_SC_FC/482 : SUCCESS in 2h 52m 49s\n- Dell_SC_iSCSI http://publiclogs.emc.com/621465/10/Dell_SC_iSCSI/471 : SUCCESS in 3h 01m 56s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":10},{"id":"33f1dbb584524c859e54ba65cb52f64cc8270b9a","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-10 03:13:03.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f2dc4acc84d088d7397707b789f8a6409b5962b4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-10 03:13:41.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-cinder-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"9137eed2464f7092ea3435e955d79ec89ec448e7","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-10 03:28:07.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/11/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/ee4285b : FAILURE in 14m 43s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/11/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a460963 : FAILURE in 14m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":11},{"id":"249944f10d4c5bcd3a64ff80e38ca75a0f6bcccb","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-10 03:29:42.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/543 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/711 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"25a74833b93db60be3c196702f201f0053507491","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-10 03:56:18.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/11/check/dsvm-tempest-kvm/a7f1f08 : FAILURE in 1m 27s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/11/check/dsvm-tempest-exe/7f9ab02 : FAILURE in 39m 13s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"5673df823da4d334c3c11c38eaf5db41cc052adb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-10 04:10:21.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/11/check/fujitsu-eternus-dx-iscsi/3076d0f : FAILURE in 53m 37s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/11/check/fujitsu-eternus-dx-fc/91cc67e : FAILURE in 54m 44s","accounts_in_message":[],"_revision_number":11},{"id":"d405c4b24905d4a8f85a79c8a9ebb03e0622e75b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-10 04:31:34.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/11/check/tempest-dsvm-full-xenial/938dc79/ : SUCCESS in 1h 09m 28s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/11/check/devstack-plugin-ceph-dsvm-xenial/cd1eaca/ : SUCCESS in 1h 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"cc954c1206d8d27535b5a8d4f8631c08bdc00b17","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-10 04:36:12.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/11/iscsi/ : FAILURE in 1h 22m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/11/smb/ : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":11},{"id":"b40c1d2f4ef96bcacf6a9e345b802d12ebf7a12b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-10 05:11:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/11/check/openstack-tox-lower-constraints/31233b0/ : SUCCESS in 10m 54s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/11/check/openstack-tox-pep8/0adc39b/ : SUCCESS in 7m 59s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/11/check/openstack-tox-py27/d01f312/ : SUCCESS in 8m 28s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/11/check/openstack-tox-py35/a3b88a3/ : SUCCESS in 10m 43s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/11/check/openstack-tox-py36/fe1af8c/ : SUCCESS in 9m 26s\n- openstack-tox-docs http://logs.openstack.org/65/621465/11/check/openstack-tox-docs/a3c02b7/html/ : SUCCESS in 7m 07s\n- tempest-full http://logs.openstack.org/65/621465/11/check/tempest-full/f6dc2db/ : SUCCESS in 1h 23m 08s\n- neutron-grenade http://logs.openstack.org/65/621465/11/check/neutron-grenade/5710322/ : SUCCESS in 55m 02s\n- grenade-py3 http://logs.openstack.org/65/621465/11/check/grenade-py3/7765b5d/ : SUCCESS in 50m 44s\n- tempest-full-py3 http://logs.openstack.org/65/621465/11/check/tempest-full-py3/08ec53c/ : SUCCESS in 1h 14m 25s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/11/check/build-openstack-releasenotes/4426e59/html/ : SUCCESS in 11m 56s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/11/check/cinder-tempest-dsvm-lvm-lio-barbican/c346ecc/ : SUCCESS in 1h 55m 36s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/11/check/cinder-tox-bandit-baseline/06c7130/ : SUCCESS in 6m 39s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/11/check/nova-multiattach/9fa2472/ : SUCCESS in 1h 07m 45s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/11/check/openstack-tox-functional-py35/4ff7901/ : SUCCESS in 4m 29s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/11/check/openstack-tox-functional-py36/87acf78/ : SUCCESS in 4m 38s\n- cinder-rally-task http://logs.openstack.org/65/621465/11/check/cinder-rally-task/e4be7a3/ : SUCCESS in 36m 35s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/11/check/openstack-tox-pylint/a6a0ac7/ : FAILURE in 3m 52s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/11/check/devstack-plugin-ceph-tempest/2bddf84/ : FAILURE in 1h 32m 31s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/11/check/openstack-tox-functional/a6ff00a/ : SUCCESS in 4m 28s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/11/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/2f51084/ : SUCCESS in 53m 03s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/11/check/legacy-tempest-dsvm-lvm-multibackend/c3a623a/ : SUCCESS in 1h 14m 06s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/11/check/legacy-tempest-dsvm-full-drbd-devstack/bf2730b/ : SUCCESS in 1h 12m 24s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/11/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/1eb1d63/ : SUCCESS in 1h 17m 24s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/11/check/tempest-slow/885eea4/ : SUCCESS in 1h 44m 29s","accounts_in_message":[],"_revision_number":11},{"id":"9d4a252c5bef3273ba948033d07d9114891ecd6e","author":{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},"date":"2018-12-10 05:36:59.000000000","message":"Patch Set 11:\n\n(8 comments)","accounts_in_message":[],"_revision_number":11},{"id":"43432b78310c53dbf451b6d9e140cc317b2dc7cc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-10 06:39:28.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/11/check/dsvm-tempest-cisco-zonemanager-job/7a64a22/ : RETRY_LIMIT in 3m 46s","accounts_in_message":[],"_revision_number":11},{"id":"4d94f9f631b68b345746362a6c0963db03fd4a7c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-10 07:01:13.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"bd454f684af9146ef1dd6ec783506d4ccde368e7","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-10 07:07:50.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/11/check/synology-dsm-cinder-iscsi/da8f373 : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":11},{"id":"b0f08a1dbd5e7c0f7f7e5ab91194c6d3004f9cee","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-10 07:28:33.000000000","message":"Patch Set 11:\n\n(7 comments)","accounts_in_message":[],"_revision_number":11},{"id":"94900ed3967f6946581a1aafd67f7b5970223e81","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-10 07:43:14.000000000","message":"Patch Set 11:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-11 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":11},{"id":"ac20397c95fc8c8c31f08e0290155f78e97ad689","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-10 08:48:13.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c2ccf5e : SUCCESS in 2h 01m 28s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/11/check/3par-fc-driver-master-client-pip-ssa02-dsvm/2d02a31 : SUCCESS in 2h 01m 44s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/11/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/4666018 : SUCCESS in 1h 51m 08s","accounts_in_message":[],"_revision_number":11},{"id":"5ab619d427984a5abf18151bb46a5c19c3030788","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-10 12:10:22.000000000","message":"Patch Set 11:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/11/check/dsvm-veritas-access-cinder-driver/d399ba2 : FAILURE in 1h 39m 07s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/11/check/dsvm-veritas-access-cinder-iscsi-driver/91978d0 : FAILURE in 1h 39m 36s","accounts_in_message":[],"_revision_number":11},{"id":"7500cbe436172f7eed9ce526500477b3c8e827ca","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-10 12:25:33.000000000","message":"Patch Set 11:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-11/nedge_iscsi : FAILURE in 26m 47s","accounts_in_message":[],"_revision_number":11},{"id":"8bf24b59c7ddfba66ed385e433d80c940f0d8185","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-10 12:31:24.000000000","message":"Patch Set 10:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/20ae85f : FAILURE in 51m 49s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/10/check/3par-fc-driver-master-client-pip-ssa02-dsvm/1c34a9d : SUCCESS in 1h 59m 08s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/10/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/a110508 : SUCCESS in 2h 03m 08s","accounts_in_message":[],"_revision_number":10},{"id":"a7186a0ca9d41e981e05c0c386127e333ba81451","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-10 13:03:07.000000000","message":"Patch Set 11:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-11/ns5_iscsi : SUCCESS in 37m 24s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-11/ns5_nfs : SUCCESS in 39m 1s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-11/ns_nfs : SUCCESS in 65m 7s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-11/ns_iscsi : SUCCESS in 55m 9s","accounts_in_message":[],"_revision_number":11},{"id":"7b643a4b1df7c4b53ae599629457274d3797340d","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-10 13:23:49.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/11/check/NEC-ISCSI/53ebd6d : FAILURE in 1h 02m 04s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/11/check/NEC-FC/9cd4c12 : FAILURE in 46m 22s","accounts_in_message":[],"_revision_number":11},{"id":"e1d9101b702f407027974d882eeec1a8fbc43524","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-10 14:37:35.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/11/check/dsvm-tempest-storpool/ca4cef4 : SUCCESS in 1h 39m 47s","accounts_in_message":[],"_revision_number":11},{"id":"21ce54c4f5507db3588749529dfcce53a3ca2632","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-10 14:44:12.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/11/check/EMC_XtremIO_FC/3a04b73/EMC_XtremIO_FC/987 : SUCCESS in 1h 32m 19s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/11/check/EMC_XtremIO_iSCSI/e50691c/EMC_XtremIO_iSCSI/1008 : SUCCESS in 1h 22m 57s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":11},{"id":"f8abe1150756ab5827baba874e0a952b96659413","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-10 14:54:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"46c3e13b887cb0fb865e1f67a45e0d01528bffc9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-10 14:55:39.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/12/check/dsvm-tempest-infinibox-fc/cd597ce : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/12/check/dsvm-tempest-infinibox-iscsi/c4ce091 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":12},{"id":"fc9252325769e887c8d9163259fc9c129caabcc9","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-10 15:02:37.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f1e485ccbfc5191dabde31fa11dd9fa77cd9e29d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-10 15:03:33.000000000","message":"Patch Set 13:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/13/check/dsvm-tempest-infinibox-fc/5f2b50c : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/13/check/dsvm-tempest-infinibox-iscsi/df70803 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":13},{"id":"517959ea8bcf55c7a270acdc0091af8448abe463","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-10 15:10:52.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/551 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/719 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"10d7f2f95620776b5ff7629ddf9d202dc393e482","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-10 15:18:30.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/13/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1d6a95d : FAILURE in 15m 35s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/13/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/adac7ae : FAILURE in 14m 41s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":13},{"id":"2e8c09df158e8a02f18f59fd5e23a72d21bd0964","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-10 15:19:01.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"78cddfd6594825c810edebfcae813e8b5f00a1ac","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-10 15:27:12.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/552 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/720 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"be5a1f76d82fbe5b9b9b752c9f43247ceb401a78","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-10 15:39:11.000000000","message":"Patch Set 13:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/13/check-not-vote/ext-cinder-zuul/1371265 : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":13},{"id":"3877963cbcd0a49df0fc8547753307901b53748d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-10 15:55:18.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/13/iscsi/ : FAILURE in 52m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/13/smb/ : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":13},{"id":"4abe8a8f744be16c45c0899f18d5d303becf858d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-10 16:08:05.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"926aa79c1fa3859c03441ee726a5cab9983f181d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-10 16:08:50.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/13/check/dsvm-tempest-kvm/49563ee : SUCCESS in 1h 01m 50s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/13/check/dsvm-tempest-exe/05963ef : SUCCESS in 1h 03m 46s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"1c3a9d21535684888897cc02ca8c1b93be82e1b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-10 16:15:39.000000000","message":"Patch Set 13:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/13/check/tempest-dsvm-full-xenial/9ccc6e8/ : SUCCESS in 1h 10m 07s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/13/check/devstack-plugin-ceph-dsvm-xenial/2b83092/ : SUCCESS in 1h 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"19436e9590fed53e06d77f1bd21766a626d65ff3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-10 16:29:33.000000000","message":"Patch Set 13:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/13/check/fujitsu-eternus-dx-iscsi/359fe42 : SUCCESS in 1h 23m 35s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/13/check/fujitsu-eternus-dx-fc/1e5aaee : SUCCESS in 1h 26m 15s","accounts_in_message":[],"_revision_number":13},{"id":"800df59773b99f4ca1058bc66b9f16ff1f1e4aa9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-10 17:48:37.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/13/check/openstack-tox-lower-constraints/a72b74c/ : SUCCESS in 15m 33s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/13/check/openstack-tox-pep8/7f718c4/ : SUCCESS in 7m 25s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/13/check/openstack-tox-py27/2706ca9/ : SUCCESS in 9m 01s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/13/check/openstack-tox-py35/3f6dab9/ : SUCCESS in 8m 54s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/13/check/openstack-tox-py36/1530c99/ : SUCCESS in 8m 20s\n- openstack-tox-docs http://logs.openstack.org/65/621465/13/check/openstack-tox-docs/4416b78/html/ : SUCCESS in 8m 00s\n- tempest-full http://logs.openstack.org/65/621465/13/check/tempest-full/b2e8eab/ : SUCCESS in 1h 36m 20s\n- neutron-grenade http://logs.openstack.org/65/621465/13/check/neutron-grenade/4f9b621/ : SUCCESS in 1h 03m 50s\n- grenade-py3 http://logs.openstack.org/65/621465/13/check/grenade-py3/d02d468/ : SUCCESS in 56m 19s\n- tempest-full-py3 http://logs.openstack.org/65/621465/13/check/tempest-full-py3/1e5ff70/ : SUCCESS in 1h 14m 02s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/13/check/build-openstack-releasenotes/a7a27ea/html/ : SUCCESS in 12m 09s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/13/check/cinder-tempest-dsvm-lvm-lio-barbican/9f6f837/ : SUCCESS in 1h 46m 08s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/13/check/cinder-tox-bandit-baseline/04d9300/ : SUCCESS in 7m 35s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/13/check/nova-multiattach/f2050d9/ : SUCCESS in 1h 07m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/13/check/openstack-tox-functional-py35/bc66496/ : SUCCESS in 4m 48s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/13/check/openstack-tox-functional-py36/f4b9010/ : SUCCESS in 4m 58s\n- cinder-rally-task http://logs.openstack.org/65/621465/13/check/cinder-rally-task/f8df2aa/ : SUCCESS in 38m 45s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/13/check/openstack-tox-pylint/55da65e/ : FAILURE in 5m 05s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/13/check/devstack-plugin-ceph-tempest/1a4b4dd/ : FAILURE in 1h 56m 08s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/13/check/openstack-tox-functional/6327f55/ : SUCCESS in 4m 46s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/13/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/86beaa9/ : SUCCESS in 1h 10m 47s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/13/check/legacy-tempest-dsvm-lvm-multibackend/50f07a3/ : SUCCESS in 1h 20m 47s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/13/check/legacy-tempest-dsvm-full-drbd-devstack/b9a0449/ : SUCCESS in 1h 12m 19s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/13/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/6c0a3ae/ : SUCCESS in 1h 38m 37s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/13/check/tempest-slow/d30c9b1/ : SUCCESS in 2h 08m 28s","accounts_in_message":[],"_revision_number":13},{"id":"c6ba818348bc5c0809f1524f985aca46e08a4155","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-10 18:10:07.000000000","message":"Patch Set 13:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-13/nedge_iscsi : FAILURE in 6m 6s","accounts_in_message":[],"_revision_number":13},{"id":"59e923da7df266559ce11bb175acd32a753fea3f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-10 18:23:57.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.57.58.96/logs/65/621465/13/upstream-solidfire-check/cinder-solidfire-iSCSI/ef861b1/ : SUCCESS in 34m 12s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":13},{"id":"aa405589f23a87b90effeccb7b61d04f40b4b9d8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-10 18:32:32.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.57.58.96/logs/65/621465/13/upstream-check/cinder-cDOT-iSCSI/99edd6a/ : SUCCESS in 28m 10s\n- cinder-cDOT-NFS http://13.57.58.96/logs/65/621465/13/upstream-check/cinder-cDOT-NFS/0101c85/ : SUCCESS in 28m 04s\n- cinder-cDOT-FCP http://13.57.58.96/logs/65/621465/13/upstream-check/cinder-cDOT-FCP/cbba0b9/ : SUCCESS in 43m 19s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"0d9b0102100eefe89325f94f515392bdc1cece37","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-10 18:50:19.000000000","message":"Patch Set 13:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-13 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":13},{"id":"39feacea72c2f1f3c4baf64d94624fb3b5adabe8","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-10 19:12:11.000000000","message":"Patch Set 13:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/13/check/synology-dsm-cinder-iscsi/2574a5b : FAILURE in 46m 20s","accounts_in_message":[],"_revision_number":13},{"id":"36c52a465d2f6af04f2a56c144ba83cc67b4f332","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-12-10 19:31:12.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.57.58.96/logs/65/621465/13/upstream-eseries-check/cinder-eseries-iSCSI/2be3374/ : FAILURE in 1h 41m 41s (non-voting)\n- cinder-eseries-FCP http://13.57.58.96/logs/65/621465/13/upstream-eseries-check/cinder-eseries-FCP/51c3bf3/ : SUCCESS in 1h 25m 29s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":13},{"id":"747873e68bb337d3632042c663d9b5b78a3c1242","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-10 19:41:37.000000000","message":"Patch Set 11:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/11/Lenovo-FC : SUCCESS in 36:55.25\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/11/Lenovo-iSCSI : SUCCESS in 37:26.19","accounts_in_message":[],"_revision_number":11},{"id":"00b3831208ac8c02c59eac8149911097beae3894","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-10 19:52:56.000000000","message":"Patch Set 11:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/11/HPMSA-FC : SUCCESS in 38:18.54\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/11/HPMSA-iSCSI : SUCCESS in 38:44.16","accounts_in_message":[],"_revision_number":11},{"id":"8d4a1d414c99dc3fe3774729cbdf4ca5f4f81311","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-10 20:08:05.000000000","message":"Patch Set 13:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ef8f92d : FAILURE in 1h 32m 53s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/13/check/3par-fc-driver-master-client-pip-ssa02-dsvm/ef5aa08 : FAILURE in 1h 27m 19s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/13/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/dd7abbe : SUCCESS in 1h 49m 25s","accounts_in_message":[],"_revision_number":13},{"id":"454efec22c565dc09adca1703c8ba3724ebbdca4","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-10 20:10:05.000000000","message":"Patch Set 13:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-13-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"74de03b1553c8f6bdfe100926d33bf02cef8f9d6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-10 22:07:35.000000000","message":"Patch Set 13:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-13/ns5_iscsi : SUCCESS in 50m 7s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-13/ns5_nfs : SUCCESS in 52m 5s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-13/ns_nfs : SUCCESS in 79m 26s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-13/ns_iscsi : SUCCESS in 69m 45s","accounts_in_message":[],"_revision_number":13},{"id":"435f7b8e75b19838f515570a437ecb833d2470b3","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-11 00:31:17.000000000","message":"Patch Set 13:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/13/check/dsvm-veritas-access-cinder-driver/9ecc73b : SUCCESS in 1h 28m 08s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/13/check/dsvm-veritas-access-cinder-iscsi-driver/e5cedba : FAILURE in 1h 15m 24s","accounts_in_message":[],"_revision_number":13},{"id":"12f292ef0a115dfb99d1419018bba82efd4e23b3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-11 05:07:42.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/13/check/NEC-ISCSI/9ef763a : SUCCESS in 1h 54m 44s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/13/check/NEC-FC/4ec08ee : SUCCESS in 1h 52m 26s","accounts_in_message":[],"_revision_number":13},{"id":"b062589f79248eb47bde6779d88d7fccc1accd4a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-11 06:41:59.000000000","message":"Patch Set 13:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/142a5bb : FAILURE in 47m 32s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/13/check/3par-fc-driver-master-client-pip-ssa02-dsvm/0676d84 : FAILURE in 42m 16s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/13/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/a42c556 : FAILURE in 38m 41s","accounts_in_message":[],"_revision_number":13},{"id":"0456daa03f078a5b6e4aa7b8466609aaea45c4f1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-11 07:15:18.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/13/EMC_VxFlexOS/613 : SUCCESS in 1h 48m 42s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":13},{"id":"4550b94d18a2cdc6da2dd3d509351d1a32587e7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-11 07:18:16.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/13/check/dsvm-tempest-storpool/5fc4b86 : SUCCESS in 1h 38m 41s","accounts_in_message":[],"_revision_number":13},{"id":"a184a75ebcc24eb8ec60de412163bdeb3d3b9c21","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-11 08:58:39.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/13/check/kaminario-dsvm-tempest-full-iscsi/a00161c : SUCCESS in 1h 52m 57s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/13/check/kaminario-dsvm-tempest-full-FC/176647b : SUCCESS in 1h 55m 10s","accounts_in_message":[],"_revision_number":13},{"id":"7063aba9e7e1dc55d1d74fc49facdc5f05a8a56b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-11 09:13:14.000000000","message":"Patch Set 13:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/13/Lenovo-FC : SUCCESS in 35:51.82\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/13/Lenovo-iSCSI : SUCCESS in 36:20.28","accounts_in_message":[],"_revision_number":13},{"id":"d53ff8ab8e587a5197c597745c7ea8f9ce08873f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-11 09:33:50.000000000","message":"Patch Set 13:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/13/HPMSA-FC : SUCCESS in 37:20.11\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/13/HPMSA-iSCSI : SUCCESS in 36:52.62","accounts_in_message":[],"_revision_number":13},{"id":"e54f3df884b89254ed3102f95239e4a7a63c5c28","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-11 12:30:47.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f81f254 : SUCCESS in 1h 57m 34s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/13/check/3par-fc-driver-master-client-pip-ssa02-dsvm/15e1f9b : SUCCESS in 1h 55m 12s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/13/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/0dec5d7 : SUCCESS in 1h 53m 05s","accounts_in_message":[],"_revision_number":13},{"id":"8aee959b932bde490d1536d290e3c5a61cf4f3de","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-11 16:57:03.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/13/check/EMC_XtremIO_FC/c0f18e8/EMC_XtremIO_FC/1000 : FAILURE in 1h 56m 21s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/13/check/EMC_XtremIO_iSCSI/cef1574/EMC_XtremIO_iSCSI/1021 : FAILURE in 1h 53m 35s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":13},{"id":"69d7ce89c56d10af3fb28795f445d9fc71b975a0","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-12 05:10:36.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d35ce2aac972402e0697baa14a42b14a14bfe0c3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-12 05:11:15.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/14/check/dsvm-tempest-infinibox-fc/255894f : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/14/check/dsvm-tempest-infinibox-iscsi/08fae84 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":14},{"id":"b3ccc32af91795f9be6b432d9da2dc1dfc51855c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-12 05:13:11.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"390042e478d34f4d0f97c84f5b42c249ac71f4b5","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-12 05:24:28.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/566 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/734 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"841c9670aebfef41d4d8ed7490534f7e8eff57de","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-12 05:25:53.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/14/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/4e56fcf : FAILURE in 14m 37s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/14/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/ee7c6e1 : FAILURE in 14m 57s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":14},{"id":"41a67302f61b69a3bf4c98f8864e9a22d21b2cdc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-12 05:33:56.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/14/check/tempest-dsvm-full-xenial/34ca628/ : FAILURE in 21m 05s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/14/check/devstack-plugin-ceph-dsvm-xenial/11fb7ba/ : FAILURE in 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"452fc37475ad28694c09af3b61a5c02a5f88d0aa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-12 05:44:43.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/14/check-not-vote/ext-cinder-zuul/e3d2e59 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":14},{"id":"c6116c92622fc345b32f747ab346315ad41dd0cc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-12 06:00:10.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/14/iscsi/ : FAILURE in 49m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/14/smb/ : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":14},{"id":"4ba31ab57527cad0ffe47b47299417f74a5ce618","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-12 06:04:50.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/14/check/dsvm-tempest-kvm/f709075 : SUCCESS in 53m 34s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/14/check/dsvm-tempest-exe/0474f80 : SUCCESS in 45m 45s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":14},{"id":"d403095859bc2893387fa6f5c1df1561ea8ef72f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-12 06:29:20.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"9d02fe3b752c3575f4bf5154b4ffe6cdb69e5eb5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-12 06:32:03.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/14/check-cinder/Cinder-ISER-ISCSI/744cd0e : SUCCESS in 1h 11m 03s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/14/check-cinder/Cinder-ISER-LIO/514f471 : SUCCESS in 1h 20m 17s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":14},{"id":"edc767c90f1942d5c3deb55a665c541dba442d72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-12 07:16:48.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/14/check/openstack-tox-lower-constraints/60a6b93/ : FAILURE in 14m 58s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/14/check/openstack-tox-pep8/bf7900d/ : SUCCESS in 9m 05s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/14/check/openstack-tox-py27/cbcbdc4/ : SUCCESS in 9m 24s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/14/check/openstack-tox-py35/79979bd/ : SUCCESS in 11m 19s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/14/check/openstack-tox-py36/2e8977b/ : SUCCESS in 7m 34s\n- openstack-tox-docs http://logs.openstack.org/65/621465/14/check/openstack-tox-docs/017ff4f/html/ : SUCCESS in 9m 33s\n- requirements-check http://logs.openstack.org/65/621465/14/check/requirements-check/3b400d1/ : SUCCESS in 3m 58s\n- tempest-full http://logs.openstack.org/65/621465/14/check/tempest-full/5c2ace6/ : SUCCESS in 1h 43m 43s\n- neutron-grenade http://logs.openstack.org/65/621465/14/check/neutron-grenade/5415799/ : SUCCESS in 55m 49s\n- grenade-py3 http://logs.openstack.org/65/621465/14/check/grenade-py3/8bb7c08/ : SUCCESS in 1h 15m 53s\n- tempest-full-py3 http://logs.openstack.org/65/621465/14/check/tempest-full-py3/bbdae94/ : SUCCESS in 1h 27m 12s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/14/check/build-openstack-releasenotes/730175f/html/ : SUCCESS in 12m 27s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/14/check/cinder-tempest-dsvm-lvm-lio-barbican/db55e91/ : SUCCESS in 1h 46m 25s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/14/check/cinder-tox-bandit-baseline/2aba6c6/ : SUCCESS in 8m 40s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/14/check/nova-multiattach/148ac94/ : SUCCESS in 1h 07m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/14/check/openstack-tox-functional-py35/93f6092/ : SUCCESS in 4m 36s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/14/check/openstack-tox-functional-py36/95e6d3f/ : SUCCESS in 7m 06s\n- cinder-rally-task http://logs.openstack.org/65/621465/14/check/cinder-rally-task/e4b8866/ : SUCCESS in 31m 34s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/14/check/openstack-tox-pylint/227be48/ : SUCCESS in 4m 26s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/14/check/devstack-plugin-ceph-tempest/e4a4d18/ : SUCCESS in 2h 01m 40s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/14/check/openstack-tox-functional/ae5ceb9/ : SUCCESS in 4m 18s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/14/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/7d6881a/ : SUCCESS in 56m 05s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/14/check/legacy-tempest-dsvm-lvm-multibackend/fbe8189/ : SUCCESS in 1h 22m 20s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/14/check/legacy-tempest-dsvm-full-drbd-devstack/fb59b0d/ : SUCCESS in 1h 19m 41s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/14/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/f8e5f49/ : SUCCESS in 1h 14m 00s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/14/check/tempest-slow/a56523a/ : SUCCESS in 1h 41m 56s","accounts_in_message":[],"_revision_number":14},{"id":"cbdd92427ae5854f46b46a193e464a62983b3e92","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-12-12 07:32:36.000000000","message":"Patch Set 14:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/14/check/Cinder-AS13000/cabf006 : FAILURE in 2m 29s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/14/check/cinder-test-iscsi/fb02735 : SUCCESS in 1h 40m 11s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/14/check/cinder-test-fc/ff26a76 : FAILURE in 6m 43s","accounts_in_message":[],"_revision_number":14},{"id":"160c0fc2a126775f73ac636936f16b580dad0068","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-12 08:13:46.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"fc96b0bf5e3247d589ee0e7beb17663b0eedc5f6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-12 08:14:32.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/15/check/dsvm-tempest-infinibox-fc/b1171b4 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/15/check/dsvm-tempest-infinibox-iscsi/903e8ee : NOT_REGISTERED","accounts_in_message":[],"_revision_number":15},{"id":"e8f877aa30f63eca70231694ba3fb13bad5264c9","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-12 08:28:02.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/571 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/739 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"0ada0c78abc78697410fe00c8d406b84cc5b2f7b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-12 08:28:53.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/15/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/4fcdc96 : FAILURE in 14m 49s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/15/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/4b0f4c7 : FAILURE in 14m 41s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":15},{"id":"7537798789a218b44e0ebd695dc9b6e9a9044948","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-12 08:32:18.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/15/iscsi/ : FAILURE in 7m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/15/smb/ : FAILURE in 10m 05s","accounts_in_message":[],"_revision_number":15},{"id":"6484a29bc35e02b692bc81a5c63fd15bdc9d2d45","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-12 09:12:50.000000000","message":"Patch Set 15:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/15/check/fujitsu-eternus-dx-iscsi/344f142 : FAILURE in 17m 11s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/15/check/fujitsu-eternus-dx-fc/a28e833 : FAILURE in 50m 35s","accounts_in_message":[],"_revision_number":15},{"id":"e7c1ee285c2880571c23fd7114bd87ea7bbed43e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-12 09:19:42.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/15/check/tempest-dsvm-full-xenial/7ed37e0/ : FAILURE in 23m 22s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/15/check/devstack-plugin-ceph-dsvm-xenial/b8c459d/ : FAILURE in 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0bf2041921e009779809c0273a6b4ddca330243a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-12 10:00:34.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/15/check/dsvm-tempest-kvm/3a0f01d : SUCCESS in 1h 01m 38s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/15/check/dsvm-tempest-exe/9e22667 : SUCCESS in 45m 37s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"a7fc49e65a1496d11c032a6bde10fe28c2160b30","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-12 10:03:11.000000000","message":"Patch Set 15:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"b511cbaa8446cf6d3f1a05a280d833a56c579115","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-12 11:11:37.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/15/check/openstack-tox-lower-constraints/8895c95/ : SUCCESS in 9m 29s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/15/check/openstack-tox-pep8/a6f678c/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/15/check/openstack-tox-py27/6293e4d/ : SUCCESS in 10m 01s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/15/check/openstack-tox-py35/528d5c9/ : SUCCESS in 9m 08s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/15/check/openstack-tox-py36/377f471/ : SUCCESS in 9m 05s\n- openstack-tox-docs http://logs.openstack.org/65/621465/15/check/openstack-tox-docs/0d4ac2f/html/ : SUCCESS in 8m 54s\n- requirements-check http://logs.openstack.org/65/621465/15/check/requirements-check/0ad85e8/ : SUCCESS in 2m 44s\n- tempest-full http://logs.openstack.org/65/621465/15/check/tempest-full/7e4a5b3/ : SUCCESS in 1h 41m 58s\n- neutron-grenade http://logs.openstack.org/65/621465/15/check/neutron-grenade/d26bab5/ : SUCCESS in 57m 54s\n- grenade-py3 http://logs.openstack.org/65/621465/15/check/grenade-py3/43811fa/ : SUCCESS in 1h 02m 51s\n- tempest-full-py3 http://logs.openstack.org/65/621465/15/check/tempest-full-py3/0274a9a/ : SUCCESS in 1h 22m 08s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/15/check/build-openstack-releasenotes/0f1065b/html/ : SUCCESS in 16m 10s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/15/check/cinder-tempest-dsvm-lvm-lio-barbican/3bab3fd/ : SUCCESS in 2h 02m 36s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/15/check/cinder-tox-bandit-baseline/01d3b66/ : SUCCESS in 6m 42s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/15/check/nova-multiattach/e8c0bba/ : SUCCESS in 53m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/15/check/openstack-tox-functional-py35/9c91bc0/ : SUCCESS in 4m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/15/check/openstack-tox-functional-py36/dd26a0d/ : SUCCESS in 5m 29s\n- cinder-rally-task http://logs.openstack.org/65/621465/15/check/cinder-rally-task/dc0a035/ : SUCCESS in 39m 19s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/15/check/openstack-tox-pylint/246c8ec/ : FAILURE in 4m 11s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/15/check/devstack-plugin-ceph-tempest/c0a52b7/ : SUCCESS in 1h 34m 35s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/15/check/openstack-tox-functional/951e46e/ : SUCCESS in 5m 28s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/15/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/74f9da7/ : SUCCESS in 59m 58s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/15/check/legacy-tempest-dsvm-lvm-multibackend/a0845c4/ : SUCCESS in 1h 07m 40s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/15/check/legacy-tempest-dsvm-full-drbd-devstack/4c1ad4d/ : SUCCESS in 1h 09m 44s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/15/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/524add7/ : SUCCESS in 1h 47m 28s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/15/check/tempest-slow/7c5af2a/ : SUCCESS in 1h 40m 32s","accounts_in_message":[],"_revision_number":15},{"id":"2fe030ee32343ddb9a74acb2239190f48d0adc3b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-12 11:48:17.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.57.58.96/logs/65/621465/15/upstream-solidfire-check/cinder-solidfire-iSCSI/6ee046d/ : SUCCESS in 35m 18s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"e8eb6542571b54b4e0f866725d7cc6d2d131089e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-12 12:11:40.000000000","message":"Patch Set 15:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-15 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":15},{"id":"48f982f883b7095a6ea10290380ad58d1045275b","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-12 12:35:40.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/15/check/synology-dsm-cinder-iscsi/167410a : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":15},{"id":"4783c3dd552c0e41d30eb3506db09c22599d6dd3","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-12 13:24:14.000000000","message":"Patch Set 15:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/15/check/dsvm-veritas-access-cinder-driver/3c61885 : FAILURE in 12s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/15/check/dsvm-veritas-access-cinder-iscsi-driver/3a0ce44 : SUCCESS in 1h 28m 38s","accounts_in_message":[],"_revision_number":15},{"id":"ffc5f71c7f15f354fca5ffa9b5a84c56b7ef491a","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-12 13:24:17.000000000","message":"Patch Set 15:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-15-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"44a873eb40b88c2af1b479a7e16aa9bafce3ec50","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-12 14:28:20.000000000","message":"Patch Set 15:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-15/ns5_iscsi : SUCCESS in 40m 54s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-15/ns5_nfs : FAILURE in 7m 20s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-15/ns_nfs : SUCCESS in 71m 28s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-15/ns_iscsi : SUCCESS in 62m 15s","accounts_in_message":[],"_revision_number":15},{"id":"a8f4fa74f7a85b87938c9557e2ba268428ce5a65","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-12-12 15:07:45.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.57.58.96/logs/65/621465/15/upstream-eseries-check/cinder-eseries-iSCSI/1753aac/ : FAILURE in 2h 09m 18s (non-voting)\n- cinder-eseries-FCP http://13.57.58.96/logs/65/621465/15/upstream-eseries-check/cinder-eseries-FCP/1e13fbe/ : SUCCESS in 1h 35m 41s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":15},{"id":"cb916dd937894e772e046668acb90393cfb96bc1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-12 15:24:14.000000000","message":"Patch Set 15:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/fe76407 : FAILURE in 1h 45m 24s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/15/check/3par-fc-driver-master-client-pip-ssa02-dsvm/57f5eae : FAILURE in 1h 32m 33s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/15/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/c88e602 : SUCCESS in 1h 56m 46s","accounts_in_message":[],"_revision_number":15},{"id":"e8727f1b35c7124da3eb8151217afd0ecdc9930a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-12 16:54:08.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.57.58.96/logs/65/621465/15/upstream-check/cinder-cDOT-iSCSI/cfe3685/ : SUCCESS in 28m 24s\n- cinder-cDOT-NFS http://13.57.58.96/logs/65/621465/15/upstream-check/cinder-cDOT-NFS/bdf96ee/ : SUCCESS in 27m 51s\n- cinder-cDOT-FCP http://13.57.58.96/logs/65/621465/15/upstream-check/cinder-cDOT-FCP/50f6ab9/ : POST_FAILURE in 17m 34s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"dd8fbbb05009bd2fc70fac4272c1c4e0780bca95","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-12 17:21:06.000000000","message":"Patch Set 15:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-15/nedge_iscsi : FAILURE in 39m 19s","accounts_in_message":[],"_revision_number":15},{"id":"ed306af691b13ddd0f9f8e773f89e9a8762f6597","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-12 18:25:04.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/15/check/EMC_Unity_FC/9ff4cd6/EMC_Unity_FC/479 : SUCCESS in 1h 38m 55s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/15/check/EMC_Unity_iSCSI/62808d3/EMC_Unity_iSCSI/457 : SUCCESS in 1h 37m 28s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"18d036c6df0b90525d7b7bc12c5b18e11fd290c3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-12 18:34:54.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/15/check/dsvm-tempest-storpool/df22ff8 : SUCCESS in 1h 39m 40s","accounts_in_message":[],"_revision_number":15},{"id":"ba67f1a82e11faa3f95ed60f3dea1884f18fc201","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-12 18:49:46.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/15/EMC_VxFlexOS/632 : SUCCESS in 1h 53m 09s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":15},{"id":"694121112fb6a69f5407acce439d0f6a246bd36c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-12 19:10:59.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/15/check/NEC-ISCSI/cdd56c3 : SUCCESS in 1h 59m 11s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/15/check/NEC-FC/92189b3 : FAILURE in 42m 48s","accounts_in_message":[],"_revision_number":15},{"id":"17107bfc535c18b079b2cc7be6a4c6b6296935ab","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-12 21:03:58.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/15/check/kaminario-dsvm-tempest-full-iscsi/8427056 : SUCCESS in 1h 49m 28s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/15/check/kaminario-dsvm-tempest-full-FC/bf42310 : SUCCESS in 1h 43m 41s","accounts_in_message":[],"_revision_number":15},{"id":"3065b53f5e568d18693b6d2794ef42f30892609b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-13 00:38:18.000000000","message":"Patch Set 15:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/15/HPMSA-FC : SUCCESS in 38:21.49\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/15/HPMSA-iSCSI : SUCCESS in 37:48.88","accounts_in_message":[],"_revision_number":15},{"id":"c1da3703bd91388e59ac9684f622c758dc7758b5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2018-12-13 04:56:09.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VNX_iSCSI http://publiclogs.emc.com/621465/15/EMC_VNX_iSCSI/289 : FAILURE in 1h 30m 05s\n- EMC_VNX_FC http://publiclogs.emc.com/621465/15/EMC_VNX_FC/258 : ABORTED in 3h 05m 14s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":15},{"id":"9d553380c2b545d33be60dfca46af180d0cbdb36","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-13 08:14:11.000000000","message":"Patch Set 15:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1828965 : FAILURE in 1h 09m 47s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/15/check/3par-fc-driver-master-client-pip-ssa02-dsvm/45b3ccf : FAILURE in 49m 44s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/15/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/0057d42 : FAILURE in 59m 32s","accounts_in_message":[],"_revision_number":15},{"id":"da3fad4dfa656b0551ba28a62c5fa3f854d69931","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-13 08:57:30.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"4abce4fbc9dde01dd49006daeb9255af8f4a5d07","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-13 08:58:11.000000000","message":"Patch Set 16:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/16/check/dsvm-tempest-infinibox-fc/d754e5d : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/16/check/dsvm-tempest-infinibox-iscsi/abb01e9 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":16},{"id":"bf45c30de0c6e250be164314b45ed7687893bd7f","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-13 08:59:04.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"4948697c5af3c7faf6bca1aca4ef5ff955ecb21c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-13 08:59:45.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/17/check/dsvm-tempest-infinibox-fc/32e9928 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/17/check/dsvm-tempest-infinibox-iscsi/76660b4 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":17},{"id":"08c4dfc3425ef3c123ac2bef9558a170688dc74e","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-13 09:11:24.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/585 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/753 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"98cabfe8202e019a18a293050da009ca3a6cbb37","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-13 09:14:16.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/17/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c61efd1 : FAILURE in 14m 54s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/17/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/42fe5c5 : FAILURE in 14m 51s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":17},{"id":"4ae5f73131295cecff2e9fe0b62ef1d1d2dad720","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-13 09:21:44.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/17/check/tempest-dsvm-full-xenial/9ebefed/ : FAILURE in 22m 21s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/17/check/devstack-plugin-ceph-dsvm-xenial/1347e27/ : FAILURE in 22m 16s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f2a5faf5dd9f91183e0a5e20d8011806e058e3c9","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-13 09:25:32.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/586 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/754 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"400bde359c8d39e1385a9d1fadda015c358d19e3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-13 09:50:01.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/17/iscsi/ : SUCCESS in 50m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/17/smb/ : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":17},{"id":"b5b042ed960bc715ba1d3db30790d3c0d5ca0f2c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-13 10:12:00.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/17/check-cinder/Cinder-ISER-ISCSI/2864b2c : SUCCESS in 57m 15s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/17/check-cinder/Cinder-ISER-LIO/29afea8 : SUCCESS in 1h 01m 01s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":17},{"id":"d6ec0dc7b526e63026c1bd0d8968ab07ecc18633","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2018-12-13 10:13:46.000000000","message":"Patch Set 17:\n\nBuild successful. \r\n* brocade_fc_zone_driver http://144.49.208.28:8000/build_logs/2018-12-13_01-15-28/ : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"e289c0980a0405871902a87ead3a2a908f7611b3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-13 10:20:29.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/17/check/dsvm-tempest-zadara-driver/d349eb3 : SUCCESS in 1h 21m 09s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"834edc58aea60af6b22ee4ea7cb9fe2957e1883f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-13 10:45:42.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/17/check/dsvm-tempest-kvm/4f6389e : SUCCESS in 1h 13m 45s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/17/check/dsvm-tempest-exe/9e4152a : SUCCESS in 1h 07m 25s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"1f0c7d1d687bf5847e568a732254129464c169da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-13 10:51:05.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/17/check/openstack-tox-lower-constraints/19443d6/ : SUCCESS in 10m 11s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/17/check/openstack-tox-pep8/bd389d0/ : SUCCESS in 7m 10s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/17/check/openstack-tox-py27/3ae34ed/ : FAILURE in 8m 49s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/17/check/openstack-tox-py35/763d83c/ : SUCCESS in 8m 16s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/17/check/openstack-tox-py36/8ce7663/ : SUCCESS in 9m 02s\n- openstack-tox-docs http://logs.openstack.org/65/621465/17/check/openstack-tox-docs/34a9c07/html/ : SUCCESS in 7m 45s\n- requirements-check http://logs.openstack.org/65/621465/17/check/requirements-check/c839f12/ : SUCCESS in 2m 47s\n- tempest-full http://logs.openstack.org/65/621465/17/check/tempest-full/4bb3990/ : FAILURE in 1h 19m 30s\n- neutron-grenade http://logs.openstack.org/65/621465/17/check/neutron-grenade/479daa2/ : SUCCESS in 57m 18s\n- grenade-py3 http://logs.openstack.org/65/621465/17/check/grenade-py3/4d1d51e/ : SUCCESS in 54m 22s\n- tempest-full-py3 http://logs.openstack.org/65/621465/17/check/tempest-full-py3/ed78d01/ : SUCCESS in 1h 13m 24s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/17/check/build-openstack-releasenotes/f00daaa/html/ : SUCCESS in 12m 52s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/17/check/cinder-tempest-dsvm-lvm-lio-barbican/d3927b9/ : SUCCESS in 1h 44m 29s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/17/check/cinder-tox-bandit-baseline/01c6f06/ : SUCCESS in 7m 09s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/17/check/nova-multiattach/67d45e7/ : SUCCESS in 1h 03m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/17/check/openstack-tox-functional-py35/41319f6/ : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/17/check/openstack-tox-functional-py36/0731b1b/ : SUCCESS in 4m 46s\n- cinder-rally-task http://logs.openstack.org/65/621465/17/check/cinder-rally-task/0d7e2b8/ : SUCCESS in 34m 08s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/17/check/openstack-tox-pylint/51a59d1/ : FAILURE in 4m 31s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/17/check/devstack-plugin-ceph-tempest/4bb1e0b/ : FAILURE in 1h 28m 49s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/17/check/openstack-tox-functional/63b6456/ : SUCCESS in 4m 54s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/17/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/53dae88/ : SUCCESS in 1h 03m 53s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-lvm-multibackend/6859c4d/ : SUCCESS in 1h 10m 01s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-full-drbd-devstack/7518ed2/ : SUCCESS in 1h 09m 39s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/347cc74/ : SUCCESS in 1h 30m 19s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/17/check/tempest-slow/b5f81b7/ : SUCCESS in 1h 34m 33s","accounts_in_message":[],"_revision_number":17},{"id":"be65957a7e74df3a038ead11b3abcd800125cc30","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-13 11:13:58.000000000","message":"Patch Set 17:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/17/check/fujitsu-eternus-dx-iscsi/c6f4538 : SUCCESS in 2h 13m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/17/check/fujitsu-eternus-dx-fc/c9bd92f : SUCCESS in 2h 10m 50s","accounts_in_message":[],"_revision_number":17},{"id":"31525e3bf03feafc9486a8380753847d638b761e","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-13 11:29:18.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"4cefb67b154304fc71c3e4bd254550d0ccf73e54","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-13 11:38:21.000000000","message":"Patch Set 17:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/17/check/dsvm-veritas-access-cinder-driver/58cd43c : FAILURE in 1h 32m 30s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/17/check/dsvm-veritas-access-cinder-iscsi-driver/8541344 : FAILURE in 1h 45m 43s","accounts_in_message":[],"_revision_number":17},{"id":"5d77258d69e1d76cbfaea4515af489bbbb5f470f","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-13 11:43:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"34c8460022634e46b36f754da4152f9ceacacf95","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-13 11:45:33.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/17/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/105f29b : FAILURE in 15m 13s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/17/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/906f9cf : FAILURE in 15m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":17},{"id":"8c01ad14c2eabe4be969b7083e03a95711011a2e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-13 11:51:23.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/17/check/tempest-dsvm-full-xenial/e2247b6/ : FAILURE in 21m 47s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/17/check/devstack-plugin-ceph-dsvm-xenial/f53f116/ : FAILURE in 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"98c7cbfc30cc8a607c0642872b2766453c17890a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-13 12:21:42.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/17/iscsi/ : SUCCESS in 51m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/17/smb/ : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":17},{"id":"59ae1276180968d7ba72d72060ebed2eebfe45a6","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-13 12:41:23.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/17/check/dsvm-tempest-kvm/9d8f677 : SUCCESS in 1h 09m 50s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/17/check/dsvm-tempest-exe/3b61b42 : FAILURE in 1h 10m 35s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"41881cea3363f2e6f84eff9b75ea2dded603daab","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-13 13:33:58.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/17/check/dsvm-tempest-zadara-driver/b05516e : FAILURE in 1h 52m 16s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"5b8098f58884378eeb3491e09dbb2c3de0e079d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-13 14:02:16.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/17/check/openstack-tox-lower-constraints/1e8dbbf/ : SUCCESS in 11m 32s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/17/check/openstack-tox-pep8/b04760e/ : SUCCESS in 7m 53s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/17/check/openstack-tox-py27/4e63745/ : SUCCESS in 8m 51s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/17/check/openstack-tox-py35/36a01f0/ : SUCCESS in 9m 31s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/17/check/openstack-tox-py36/66bc1e3/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/65/621465/17/check/openstack-tox-docs/2c22a9c/html/ : SUCCESS in 7m 45s\n- requirements-check http://logs.openstack.org/65/621465/17/check/requirements-check/8c45414/ : SUCCESS in 3m 45s\n- tempest-full http://logs.openstack.org/65/621465/17/check/tempest-full/02ddb33/ : SUCCESS in 1h 52m 17s\n- neutron-grenade http://logs.openstack.org/65/621465/17/check/neutron-grenade/d16ce8f/ : SUCCESS in 51m 07s\n- grenade-py3 http://logs.openstack.org/65/621465/17/check/grenade-py3/5d2595b/ : FAILURE in 56m 28s\n- tempest-full-py3 http://logs.openstack.org/65/621465/17/check/tempest-full-py3/c391aa1/ : SUCCESS in 1h 06m 00s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/17/check/build-openstack-releasenotes/6a418dc/html/ : SUCCESS in 12m 03s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/17/check/cinder-tempest-dsvm-lvm-lio-barbican/ec7c2d2/ : SUCCESS in 1h 51m 29s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/17/check/cinder-tox-bandit-baseline/0694ec2/ : SUCCESS in 8m 09s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/17/check/nova-multiattach/d607306/ : SUCCESS in 1h 32m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/17/check/openstack-tox-functional-py35/aca85a4/ : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/17/check/openstack-tox-functional-py36/6b222d7/ : SUCCESS in 5m 04s\n- cinder-rally-task http://logs.openstack.org/65/621465/17/check/cinder-rally-task/37a7ec0/ : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/17/check/openstack-tox-pylint/98e4d67/ : FAILURE in 4m 30s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/17/check/devstack-plugin-ceph-tempest/17e8191/ : SUCCESS in 1h 35m 22s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/17/check/openstack-tox-functional/a703ae3/ : SUCCESS in 5m 01s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/17/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/36ebac1/ : SUCCESS in 1h 00m 29s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-lvm-multibackend/505013d/ : SUCCESS in 1h 24m 52s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-full-drbd-devstack/6994507/ : SUCCESS in 1h 21m 04s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/c70d96e/ : SUCCESS in 1h 36m 00s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/17/check/tempest-slow/cf32d20/ : SUCCESS in 1h 53m 19s","accounts_in_message":[],"_revision_number":17},{"id":"655829f902594935dc5cd4d932de715ed2db9690","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-12-13 14:13:35.000000000","message":"Patch Set 17:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- grenade-py3: https://bugs.launchpad.net/bugs/1763070\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":17},{"id":"4222e3234001436683eeda1eb0163fbdc4a23baa","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-13 14:34:33.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"1a3f1d0012c366ac41704164fab2058a03c07fbe","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-13 14:49:30.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/17/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1e50754 : FAILURE in 14m 42s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/17/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a64d3df : FAILURE in 14m 32s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":17},{"id":"ad773f297d1d4acf7a7a966e7bf3f9272991ba9c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-13 14:49:59.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/17/check/tempest-dsvm-full-xenial/3fa25eb/ : FAILURE in 11m 47s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/17/check/devstack-plugin-ceph-dsvm-xenial/7a106c5/ : FAILURE in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0839d8e52e8c2edd7f8fa9f2820d7984a1248610","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-13 15:19:03.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/17/check/dsvm-tempest-storpool/fcdc2a4 : SUCCESS in 1h 41m 37s","accounts_in_message":[],"_revision_number":17},{"id":"f4cc0268e92262a1f2b589b3966d22889a66908f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-13 15:25:11.000000000","message":"Patch Set 15:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/15/Lenovo-FC : SUCCESS in 35:46.83\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/15/Lenovo-iSCSI : SUCCESS in 36:40.28","accounts_in_message":[],"_revision_number":15},{"id":"97fd2e53117e52a87256709b016ed13b14c954cb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-13 15:33:25.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/17/iscsi/ : SUCCESS in 58m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/17/smb/ : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":17},{"id":"8e3769269a2b963b9a867e64409d32bbb48e1812","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-13 15:40:28.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/17/check-not-vote/ext-cinder-zuul/2d126d2 : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":17},{"id":"8f2bf4415a1238e2df21b615be7bb60c39f1a651","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-13 16:20:06.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/17/check/EMC_Unity_FC/cc7a2bc/EMC_Unity_FC/495 : SUCCESS in 1h 20m 05s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/17/check/EMC_Unity_iSCSI/4b01bda/EMC_Unity_iSCSI/470 : SUCCESS in 1h 37m 22s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"d3410ee5ddf5f7598439082300360ee7fdc88b2a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-13 17:12:44.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/17/check/dsvm-tempest-kvm/06250dc : SUCCESS in 1h 15m 25s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/17/check/dsvm-tempest-exe/48a9e48 : SUCCESS in 1h 03m 01s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"7184c4910fd93f800a3e9d80c87aae4225610ac8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-13 17:37:34.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/17/check/NEC-ISCSI/53c384e : FAILURE in 2h 50m 14s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/17/check/NEC-FC/ddfc1ff : FAILURE in 24m 16s","accounts_in_message":[],"_revision_number":17},{"id":"1980df285bcfb736a5655128df1c84a58a30ad7f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-13 17:52:27.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/17/check/EMC_XtremIO_FC/5028cee/EMC_XtremIO_FC/1016 : FAILURE in 1h 54m 36s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/17/check/EMC_XtremIO_iSCSI/ef6cefa/EMC_XtremIO_iSCSI/1039 : FAILURE in 1h 54m 26s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":17},{"id":"15a934a78a695232226409235e083ce214d3bf85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-13 18:13:05.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/17/check/openstack-tox-lower-constraints/9291fcf/ : SUCCESS in 9m 21s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/17/check/openstack-tox-pep8/f710346/ : SUCCESS in 7m 01s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/17/check/openstack-tox-py27/e97ca66/ : SUCCESS in 10m 13s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/17/check/openstack-tox-py35/a195624/ : SUCCESS in 7m 56s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/17/check/openstack-tox-py36/5d63097/ : SUCCESS in 9m 16s\n- openstack-tox-docs http://logs.openstack.org/65/621465/17/check/openstack-tox-docs/5d7716c/html/ : SUCCESS in 7m 28s\n- requirements-check http://logs.openstack.org/65/621465/17/check/requirements-check/56ff36d/ : SUCCESS in 3m 00s\n- tempest-full http://logs.openstack.org/65/621465/17/check/tempest-full/10a2e55/ : SUCCESS in 1h 27m 47s\n- neutron-grenade http://logs.openstack.org/65/621465/17/check/neutron-grenade/363d5e1/ : SUCCESS in 55m 09s\n- grenade-py3 http://logs.openstack.org/65/621465/17/check/grenade-py3/fe4a263/ : SUCCESS in 59m 47s\n- tempest-full-py3 http://logs.openstack.org/65/621465/17/check/tempest-full-py3/cccd83d/ : SUCCESS in 1h 23m 33s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/17/check/build-openstack-releasenotes/f03296f/html/ : SUCCESS in 10m 12s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/17/check/cinder-tempest-dsvm-lvm-lio-barbican/02bf285/ : SUCCESS in 2h 11m 37s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/17/check/cinder-tox-bandit-baseline/bd50bfe/ : SUCCESS in 8m 56s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/17/check/nova-multiattach/73baa3f/ : SUCCESS in 1h 14m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/17/check/openstack-tox-functional-py35/d7ff943/ : SUCCESS in 4m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/17/check/openstack-tox-functional-py36/0cbe0bc/ : SUCCESS in 5m 11s\n- cinder-rally-task http://logs.openstack.org/65/621465/17/check/cinder-rally-task/6c659cb/ : SUCCESS in 37m 19s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/17/check/openstack-tox-pylint/01d1dd9/ : FAILURE in 4m 45s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/17/check/devstack-plugin-ceph-tempest/24a2e35/ : FAILURE in 1h 45m 33s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/17/check/openstack-tox-functional/c5ff7ab/ : SUCCESS in 4m 57s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/17/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/daee569/ : SUCCESS in 1h 04m 28s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-lvm-multibackend/0e3bfa9/ : SUCCESS in 1h 13m 35s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-full-drbd-devstack/558862f/ : SUCCESS in 1h 17m 30s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/17/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/7f07e6d/ : SUCCESS in 1h 22m 03s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/17/check/tempest-slow/f3bd33d/ : SUCCESS in 1h 35m 08s","accounts_in_message":[],"_revision_number":17},{"id":"94c160b8d91ba4bfb13e6aa7fca3bc2e80c0826a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-13 18:50:09.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.57.58.96/logs/65/621465/17/upstream-solidfire-check/cinder-solidfire-iSCSI/92baee8/ : SUCCESS in 34m 36s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":17},{"id":"fff385a77bd3781b75bc4c0d74f4e2d96d3752b1","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-13 18:53:40.000000000","message":"Patch Set 17:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-17/nedge_iscsi : SUCCESS in 39m 46s","accounts_in_message":[],"_revision_number":17},{"id":"b44eab269cf58ff93b546158ad256c49267d50c7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-13 19:01:30.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.57.58.96/logs/65/621465/17/upstream-check/cinder-cDOT-iSCSI/6b15361/ : POST_FAILURE in 2m 15s\n- cinder-cDOT-NFS http://13.57.58.96/logs/65/621465/17/upstream-check/cinder-cDOT-NFS/0b102ad/ : SUCCESS in 28m 18s\n- cinder-cDOT-FCP http://13.57.58.96/logs/65/621465/17/upstream-check/cinder-cDOT-FCP/53a8f12/ : SUCCESS in 47m 30s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":17},{"id":"5462740af8b95412b7ab58d5547a9c88ed562c39","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-13 19:26:08.000000000","message":"Patch Set 17:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-17-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"cebbc9f3528b131528ccd42d896671b29b4fa675","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-13 19:29:41.000000000","message":"Patch Set 17:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-17/ns5_iscsi : SUCCESS in 43m 45s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-17/ns5_nfs : SUCCESS in 48m 11s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-17/ns_nfs : SUCCESS in 75m 37s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-17/ns_iscsi : SUCCESS in 65m 4s","accounts_in_message":[],"_revision_number":17},{"id":"b4fcd019de2a2a767056c057f4a27d91d5328e18","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-13 19:32:46.000000000","message":"Patch Set 17:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-17 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":17},{"id":"740f711efca5fecdd685126fb1755ae6473f99c6","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-13 19:43:31.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/17/check/synology-dsm-cinder-iscsi/a4b788c : FAILURE in 47m 29s","accounts_in_message":[],"_revision_number":17},{"id":"04f259c398f6e2526743a2da9a0119ec35ff7bc4","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-14 00:05:49.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/17/check/kaminario-dsvm-tempest-full-iscsi/768cfce : SUCCESS in 1h 50m 30s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/17/check/kaminario-dsvm-tempest-full-FC/bf46436 : SUCCESS in 1h 41m 39s","accounts_in_message":[],"_revision_number":17},{"id":"b0054e590244ecf6af6829ff0aa36ffb8b14d141","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-12-14 00:36:04.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.57.58.96/logs/65/621465/17/upstream-eseries-check/cinder-eseries-iSCSI/1d110d0/ : POST_FAILURE in 1h 36m 31s (non-voting)\n- cinder-eseries-FCP http://13.57.58.96/logs/65/621465/17/upstream-eseries-check/cinder-eseries-FCP/8529d01/ : POST_FAILURE in 4m 54s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":17},{"id":"dfcfc9a98f3c40f526822c93fef91a4e1bbdd0e9","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-12-14 00:48:30.000000000","message":"Patch Set 17:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/17/check/Cinder-AS13000/8fe1bfd : FAILURE in 40m 05s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/17/check/cinder-test-iscsi/84df172 : FAILURE in 2h 00m 15s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/17/check/cinder-test-fc/60fc46e : FAILURE in 59m 28s","accounts_in_message":[],"_revision_number":17},{"id":"4f3c6e7aae22b31efdeb5fb23b7cf8875ed4e6ef","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2018-12-14 02:29:13.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VNX_iSCSI http://publiclogs.emc.com/621465/17/EMC_VNX_iSCSI/302 : FAILURE in 1h 27m 46s\n- EMC_VNX_FC http://publiclogs.emc.com/621465/17/EMC_VNX_FC/271 : SUCCESS in 1h 03m 19s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":17},{"id":"18d7698de204f928de0d890eeb46890165f1ed9b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-14 02:40:04.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/17/check/dsvm-tempest-zadara-driver/e0c34ee : FAILURE in 1h 52m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"75170049944f748a8ed07073c0dd2c0b57c98280","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-14 02:41:02.000000000","message":"Patch Set 17:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/17/HPMSA-FC : SUCCESS in 38:44.38\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/17/HPMSA-iSCSI : SUCCESS in 38:35.29","accounts_in_message":[],"_revision_number":17},{"id":"b4e6c36934a506f016e6f18dcd2fa3503eb468b9","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-14 08:06:17.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"2d7a65159818f8cc1e3b16093d2946d9b1c851c0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-14 08:07:07.000000000","message":"Patch Set 18:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/18/check/dsvm-tempest-infinibox-fc/2a46922 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/18/check/dsvm-tempest-infinibox-iscsi/a9377a9 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":18},{"id":"620bc63b489c97e7746a9a65ba215658860420ab","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-14 08:07:09.000000000","message":"Patch Set 18:\n\nBuild failed\n\n- ext-cinder-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":18},{"id":"791ad6815cf2f75f73aa0c1f9156c814427f3850","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-14 08:20:40.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/606 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/774 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"c3847bee1909ab40e7089b4eb9149a12754f27e5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-14 08:53:33.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/18/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/6101b59 : FAILURE in 17m 06s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/18/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/9e98a40 : FAILURE in 14m 13s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":18},{"id":"d204d5937d51baa9c4b7dea9c53e36ec826afcbe","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-14 09:23:00.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/18/check-cinder/Cinder-ISER-ISCSI/7a8d008 : SUCCESS in 59m 36s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/18/check-cinder/Cinder-ISER-LIO/d895315 : SUCCESS in 1h 06m 11s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":18},{"id":"b5e50cdf0f1f1593c62c8e7efe9782e5353e7c7d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-14 09:25:48.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/18/iscsi/ : FAILURE in 1h 18m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/18/smb/ : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":18},{"id":"68db8a4dc30da47e7d517f7dd96113d288ccce43","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-14 09:27:09.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/18/check/dsvm-tempest-kvm/f58cc5b : SUCCESS in 1h 19m 26s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/18/check/dsvm-tempest-exe/c808aa4 : SUCCESS in 56m 56s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"2fdc9236da16eb610c0c5514a9ff36bf440a3889","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-14 09:29:36.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/18/check/tempest-dsvm-full-xenial/f1796f0/ : FAILURE in 1h 05m 48s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/18/check/devstack-plugin-ceph-dsvm-xenial/c886776/ : SUCCESS in 1h 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"8259c0fd2be8063f6fbdb8fa858e9f0ad44b3823","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 10:11:35.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/18/check/openstack-tox-lower-constraints/b2b5cd6/ : SUCCESS in 9m 55s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/18/check/openstack-tox-pep8/ed377fb/ : SUCCESS in 7m 04s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/18/check/openstack-tox-py27/2596b19/ : SUCCESS in 9m 48s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/18/check/openstack-tox-py35/072f043/ : SUCCESS in 8m 24s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/18/check/openstack-tox-py36/272786e/ : SUCCESS in 7m 39s\n- openstack-tox-docs http://logs.openstack.org/65/621465/18/check/openstack-tox-docs/9c9c548/html/ : SUCCESS in 7m 40s\n- requirements-check http://logs.openstack.org/65/621465/18/check/requirements-check/21331db/ : SUCCESS in 2m 59s\n- tempest-full http://logs.openstack.org/65/621465/18/check/tempest-full/ab4340b/ : SUCCESS in 1h 30m 15s\n- neutron-grenade http://logs.openstack.org/65/621465/18/check/neutron-grenade/1354135/ : SUCCESS in 56m 23s\n- grenade-py3 http://logs.openstack.org/65/621465/18/check/grenade-py3/da1cc74/ : SUCCESS in 1h 02m 06s\n- tempest-full-py3 http://logs.openstack.org/65/621465/18/check/tempest-full-py3/a1ae8df/ : SUCCESS in 1h 15m 55s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/18/check/build-openstack-releasenotes/aca1b6c/html/ : SUCCESS in 13m 55s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/18/check/cinder-tempest-dsvm-lvm-lio-barbican/c96fd1a/ : SUCCESS in 1h 55m 12s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/18/check/cinder-tox-bandit-baseline/968647b/ : SUCCESS in 6m 48s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/18/check/nova-multiattach/93859c1/ : FAILURE in 1h 16m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/18/check/openstack-tox-functional-py35/318cf13/ : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/18/check/openstack-tox-functional-py36/94027b6/ : SUCCESS in 4m 58s\n- cinder-rally-task http://logs.openstack.org/65/621465/18/check/cinder-rally-task/0a8f078/ : SUCCESS in 38m 27s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/18/check/openstack-tox-pylint/d9e6091/ : FAILURE in 4m 36s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/18/check/devstack-plugin-ceph-tempest/7adeda4/ : FAILURE in 1h 47m 37s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/18/check/openstack-tox-functional/11e1fe9/ : SUCCESS in 4m 51s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/18/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/ad7ffc6/ : SUCCESS in 1h 10m 31s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/18/check/legacy-tempest-dsvm-lvm-multibackend/54b545f/ : SUCCESS in 1h 16m 19s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/18/check/legacy-tempest-dsvm-full-drbd-devstack/5a0e162/ : SUCCESS in 1h 25m 39s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/18/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/0d45ae9/ : SUCCESS in 1h 17m 54s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/18/check/tempest-slow/b62f7a3/ : SUCCESS in 1h 38m 16s","accounts_in_message":[],"_revision_number":18},{"id":"31eb48cbcbbde5c273c263a85b8c7145a18d22a6","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-12-14 10:11:47.000000000","message":"Patch Set 18:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- nova-multiattach: https://bugs.launchpad.net/bugs/1763712\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":18},{"id":"ef29855da59bcee120ae96f60bc1a83eee654086","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-14 10:24:22.000000000","message":"Patch Set 18:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/18/check/fujitsu-eternus-dx-iscsi/04b8150 : SUCCESS in 2h 14m 11s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/18/check/fujitsu-eternus-dx-fc/979b920 : SUCCESS in 2h 16m 01s","accounts_in_message":[],"_revision_number":18},{"id":"cae65ae84e6acb5a0f1035ac5fed6179296ffa0a","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-14 11:52:56.000000000","message":"Patch Set 18:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/18/check/dsvm-veritas-access-cinder-driver/c75a223 : SUCCESS in 1h 30m 41s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/18/check/dsvm-veritas-access-cinder-iscsi-driver/e1d6de4 : FAILURE in 3h 45m 54s","accounts_in_message":[],"_revision_number":18},{"id":"75c1fc53beedb71a5433333b9cff8cb240d6f674","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-14 16:59:24.000000000","message":"Patch Set 17:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/17/Lenovo-FC : SUCCESS in 36:30.47\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/17/Lenovo-iSCSI : SUCCESS in 37:34.71","accounts_in_message":[],"_revision_number":17},{"id":"3d6980c8de768a39ab5290cf65d014b6774940d0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-15 02:31:07.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/18/check/EMC_Unity_FC/5153990/EMC_Unity_FC/517 : SUCCESS in 1h 42m 39s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/18/check/EMC_Unity_iSCSI/7a7c9b8/EMC_Unity_iSCSI/490 : FAILURE in 1h 43m 16s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"2395fbd7fc6cc966296e0026e80fee15e7b75ae2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-15 04:07:24.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/18/check/dsvm-tempest-storpool/eab1008 : SUCCESS in 1h 39m 19s","accounts_in_message":[],"_revision_number":18},{"id":"fe5c0de4d8bd9c52f06fab2014988f656dbf4270","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-15 12:57:45.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/18/check/dsvm-tempest-zadara-driver/57eac9e : SUCCESS in 1h 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"86bd7d6c86b5590569e509ea5855e8defb0985b8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-15 14:00:17.000000000","message":"Patch Set 18:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/18/check/NEC-ISCSI/0e0c3ef : FAILURE in 44m 08s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/18/check/NEC-FC/a08bc5d : SUCCESS in 2h 31m 57s","accounts_in_message":[],"_revision_number":18},{"id":"9e19ec3763905842c7cd520c1d92e44b6c03843a","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-12-15 16:42:25.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/18/check/Cinder-AS13000/d3af249 : SUCCESS in 3h 17m 59s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/18/check/cinder-test-iscsi/101663e : SUCCESS in 1h 58m 02s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/18/check/cinder-test-fc/c9c0a11 : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":18},{"id":"44133bb53fcf2938baf931cd385e57ac17148638","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-15 19:23:58.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/18/check/EMC_XtremIO_FC/2f7e4bb/EMC_XtremIO_FC/1039 : FAILURE in 1h 56m 02s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/18/check/EMC_XtremIO_iSCSI/4f772ff/EMC_XtremIO_iSCSI/1062 : FAILURE in 1h 53m 57s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":18},{"id":"5ff0d4f55f665cde05d15da421cd14ac6d869a09","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-17 07:54:12.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a0bd677a5843524d7df236a41b2a603aab9157f2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-17 07:54:54.000000000","message":"Patch Set 19:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/19/check/dsvm-tempest-infinibox-fc/adb35e8 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/19/check/dsvm-tempest-infinibox-iscsi/582c2ed : NOT_REGISTERED","accounts_in_message":[],"_revision_number":19},{"id":"6edf5183cc7161af1e0e271e12c0449511ad47fd","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2018-12-17 08:11:28.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/635 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/803 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"fd1f3fe8ec0ed99ec9003609d807bb73ed0e58c5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-17 08:31:19.000000000","message":"Patch Set 19:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/19/check-not-vote/ext-cinder-zuul/98b8641 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":19},{"id":"41531e2e7bec268d3400099e4e6da5ab4e0b3f24","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-17 08:48:50.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/19/iscsi/ : SUCCESS in 53m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/19/smb/ : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":19},{"id":"bbd6bd205cda67a43380b872d30384b7c695c132","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-17 08:54:33.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/19/check/dsvm-tempest-kvm/11de377 : FAILURE in 59m 10s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/19/check/dsvm-tempest-exe/6bc1fbc : SUCCESS in 45m 24s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":19},{"id":"94bcfaeae328e87744d4209c70690b97264ae2c4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-17 08:57:04.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/19/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/96626a0 : FAILURE in 16m 57s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/19/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/4df45a0 : FAILURE in 14m 43s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":19},{"id":"9994f17a49350228acb63c6e274c6a2132b352ab","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-17 09:18:30.000000000","message":"Patch Set 19:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/19 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"1f75b0e0fb43fe3f2b7226e46c5acdabf2d90d96","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-17 09:47:28.000000000","message":"Patch Set 19:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/19/check/fujitsu-eternus-dx-iscsi/b914520 : SUCCESS in 1h 52m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/19/check/fujitsu-eternus-dx-fc/adf4ff8 : SUCCESS in 1h 47m 19s","accounts_in_message":[],"_revision_number":19},{"id":"8b7083f06292ec17ed965cba2f0d64f6f3abc419","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-17 09:51:15.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/19/check/dsvm-tempest-storpool/19d8dcc : SUCCESS in 1h 40m 19s","accounts_in_message":[],"_revision_number":19},{"id":"403a5425a86694410959fba51f2fd6e226b207e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-17 10:02:50.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/19/check/openstack-tox-lower-constraints/8b09cf9/ : SUCCESS in 14m 22s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/19/check/openstack-tox-pep8/7beb218/ : SUCCESS in 7m 36s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/19/check/openstack-tox-py27/333e22a/ : SUCCESS in 8m 00s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/19/check/openstack-tox-py35/cbd8f14/ : SUCCESS in 12m 00s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/19/check/openstack-tox-py36/30731ea/ : SUCCESS in 9m 46s\n- openstack-tox-docs http://logs.openstack.org/65/621465/19/check/openstack-tox-docs/5605372/html/ : SUCCESS in 9m 55s\n- requirements-check http://logs.openstack.org/65/621465/19/check/requirements-check/67ae537/ : SUCCESS in 5m 43s\n- tempest-full http://logs.openstack.org/65/621465/19/check/tempest-full/95f741a/ : SUCCESS in 1h 48m 44s\n- neutron-grenade http://logs.openstack.org/65/621465/19/check/neutron-grenade/98f4900/ : SUCCESS in 55m 22s\n- grenade-py3 http://logs.openstack.org/65/621465/19/check/grenade-py3/fae119a/ : SUCCESS in 1h 00m 41s\n- tempest-full-py3 http://logs.openstack.org/65/621465/19/check/tempest-full-py3/3f37e11/ : SUCCESS in 1h 10m 54s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/19/check/build-openstack-releasenotes/6d68d14/html/ : SUCCESS in 12m 21s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/19/check/cinder-tempest-dsvm-lvm-lio-barbican/3407ed7/ : SUCCESS in 1h 47m 29s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/19/check/cinder-tox-bandit-baseline/4da889f/ : SUCCESS in 7m 30s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/19/check/nova-multiattach/09dc897/ : SUCCESS in 1h 10m 41s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/19/check/openstack-tox-functional-py35/413e2cd/ : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/19/check/openstack-tox-functional-py36/85b4168/ : SUCCESS in 5m 28s\n- cinder-rally-task http://logs.openstack.org/65/621465/19/check/cinder-rally-task/5db1bfe/ : POST_FAILURE in 37m 44s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/19/check/openstack-tox-pylint/c37c928/ : FAILURE in 4m 29s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/19/check/devstack-plugin-ceph-tempest/6ea142f/ : SUCCESS in 2h 03m 14s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/19/check/openstack-tox-functional/59164e1/ : SUCCESS in 4m 44s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/19/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/dfdc331/ : SUCCESS in 1h 03m 10s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/19/check/legacy-tempest-dsvm-lvm-multibackend/e27f853/ : FAILURE in 1h 14m 03s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/19/check/legacy-tempest-dsvm-full-drbd-devstack/50289e6/ : SUCCESS in 1h 46m 14s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/19/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/52bec25/ : SUCCESS in 1h 18m 36s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/19/check/tempest-slow/c48a822/ : SUCCESS in 1h 51m 10s","accounts_in_message":[],"_revision_number":19},{"id":"cea38d637e57f1acf3b681665bb0c137f4724582","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-17 10:40:26.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.57.58.96/logs/65/621465/19/upstream-solidfire-check/cinder-solidfire-iSCSI/9dc1341/ : POST_FAILURE in 36m 28s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":19},{"id":"6439fe5fb3c70a5fb4be1a9917895f9805c2bc1a","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-17 10:44:19.000000000","message":"Patch Set 19:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-19/nedge_iscsi : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":19},{"id":"ead379888823747734703fab25cb31ad61a5f660","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-17 10:52:05.000000000","message":"Patch Set 19:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/19/check/dsvm-veritas-access-cinder-driver/2049a42 : FAILURE in 1h 36m 31s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/19/check/dsvm-veritas-access-cinder-iscsi-driver/07b5ab4 : FAILURE in 1h 36m 29s","accounts_in_message":[],"_revision_number":19},{"id":"3f9515069ef7479f72ad1ccbec4efec29ddb5c77","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-17 10:52:06.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.57.58.96/logs/65/621465/19/upstream-check/cinder-cDOT-iSCSI/0576231/ : POST_FAILURE in 29m 57s\n- cinder-cDOT-NFS http://13.57.58.96/logs/65/621465/19/upstream-check/cinder-cDOT-NFS/89ca75c/ : POST_FAILURE in 31m 02s\n- cinder-cDOT-FCP http://13.57.58.96/logs/65/621465/19/upstream-check/cinder-cDOT-FCP/83e2ebb/ : POST_FAILURE in 48m 35s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":19},{"id":"50efce5db0486021fc6fb26a7712b007fb07c151","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-17 10:54:46.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/19/check/NEC-ISCSI/1209b0a : SUCCESS in 2h 03m 18s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/19/check/NEC-FC/e34aae6 : SUCCESS in 1h 58m 25s","accounts_in_message":[],"_revision_number":19},{"id":"c857bdc712eb3317f55a7b34c60d07842c8ca3ce","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-17 10:54:51.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/19/check/synology-dsm-cinder-iscsi/3cc3d4c : SUCCESS in 51m 29s","accounts_in_message":[],"_revision_number":19},{"id":"e041fe5e340dd5ba6669eaf30fc739fc637db445","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-17 10:57:21.000000000","message":"Patch Set 19:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-19 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":19},{"id":"a584312b63b877c74fded481b05897079f2a1316","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-17 11:00:50.000000000","message":"Patch Set 19:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/19/check/tempest-dsvm-full-xenial/6b2afce/ : FAILURE in 3h 02m 35s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/19/check/devstack-plugin-ceph-dsvm-xenial/c81dd9c/ : FAILURE in 3h 02m 45s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"c82daacdc025ba43a4243a3186598167c1fb9bc4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2018-12-17 11:15:47.000000000","message":"Patch Set 19:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-19/ns5_iscsi : SUCCESS in 41m 42s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-19/ns5_nfs : SUCCESS in 44m 39s\n* nexenta-dsvm-volume-ns_nfs http://140.174.232.106/refs-changes-65-621465-19/ns_nfs : SUCCESS in 72m 46s\n* nexenta-dsvm-volume-ns_iscsi http://140.174.232.106/refs-changes-65-621465-19/ns_iscsi : SUCCESS in 61m 28s","accounts_in_message":[],"_revision_number":19},{"id":"6ec424d32260726d4aaf3f7a0bce139332f6cdb2","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-17 11:17:09.000000000","message":"Patch Set 19:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-19-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"65d8693b0d1e33252f84650b720fc49b9ca2ec63","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-17 11:29:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/19/check/dsvm-tempest-zadara-driver/1d91f71 : FAILURE in 1h 52m 36s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"415dda587e8f95d32a8ddc36a5692be95164d8c2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2018-12-17 11:40:42.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VNX_iSCSI http://publiclogs.emc.com/621465/19/EMC_VNX_iSCSI/322 : FAILURE in 1h 37m 38s\n- EMC_VNX_FC http://publiclogs.emc.com/621465/19/EMC_VNX_FC/293 : SUCCESS in 51m 42s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":19},{"id":"369343e8a447b735e5eb429c77df0eee25a20a97","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-17 11:41:16.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/19/check/EMC_Unity_FC/573ae87/EMC_Unity_FC/540 : SUCCESS in 1h 24m 01s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/19/check/EMC_Unity_iSCSI/e47f673/EMC_Unity_iSCSI/512 : FAILURE in 2h 03m 51s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"2e452ee73e15f930a611f3416ac7306c9cd5c823","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-17 11:48:02.000000000","message":"Patch Set 19:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/19/Lenovo-FC : SUCCESS in 37:06.78\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/19/Lenovo-iSCSI : SUCCESS in 37:08.35","accounts_in_message":[],"_revision_number":19},{"id":"51aa56ca6f97691a31ab619a4ab04215a058d391","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-17 11:52:29.000000000","message":"Patch Set 19:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/19/HPMSA-FC : SUCCESS in 39:01.06\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/19/HPMSA-iSCSI : SUCCESS in 38:45.90","accounts_in_message":[],"_revision_number":19},{"id":"6b62b4e2e789e21471e6e9326489018cd5ec3a7e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-17 11:54:54.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/621465/19/EMC_VxFlexOS/664 : SUCCESS in 1h 51m 45s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":19},{"id":"df10cb5d7e40d5e984112778b805bde81de5c5bc","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-12-17 12:29:44.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/19/check/Cinder-AS13000/1383d87 : SUCCESS in 2h 56m 41s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/19/check/cinder-test-iscsi/14085d1 : SUCCESS in 1h 48m 57s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/19/check/cinder-test-fc/ff648f2 : SUCCESS in 1h 17m 09s","accounts_in_message":[],"_revision_number":19},{"id":"c8174db6c2f209eb9f4c68c60604411a768504ab","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-17 12:56:26.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/19/check/EMC_XtremIO_FC/bf9cad2/EMC_XtremIO_FC/1061 : SUCCESS in 1h 31m 03s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/19/check/EMC_XtremIO_iSCSI/1de7b05/EMC_XtremIO_iSCSI/1083 : SUCCESS in 1h 27m 37s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":19},{"id":"3f86039b9f1ba5024910fa55638c976ec68acc5e","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-17 13:01:44.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/19/check/kaminario-dsvm-tempest-full-iscsi/7df3585 : SUCCESS in 1h 53m 28s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/19/check/kaminario-dsvm-tempest-full-FC/388699d : SUCCESS in 1h 45m 55s","accounts_in_message":[],"_revision_number":19},{"id":"915a14c44a7eb76a95d704731e339114172e4c63","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-12-17 13:31:31.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12289 : SUCCESS in 58m 48s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/8709 : FAILURE in 58m 06s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10493 : SUCCESS in 54m 34s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9090 : SUCCESS in 55m 19s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10056 : FAILURE in 35m 40s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1141 : FAILURE in 2h 06m 12s","accounts_in_message":[],"_revision_number":19},{"id":"847c426a867f6fc6ee588426554e281df39bfc37","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-17 15:32:34.000000000","message":"Patch Set 19:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/19/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1a66cdf : SUCCESS in 1h 55m 55s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/19/check/3par-fc-driver-master-client-pip-ssa02-dsvm/34087b2 : FAILURE in 1h 49m 39s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/19/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/36dec6b : SUCCESS in 2h 05m 35s","accounts_in_message":[],"_revision_number":19},{"id":"8f30b43add97b6bff6188e466731d0ff90b659e8","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2018-12-17 18:43:11.000000000","message":"Patch Set 19: Code-Review-1\n\nSo, we have several issues here.  There is CI missing and still the need for more unit test.  In addition, the driver is going to have to meet the minimum set of functions to be mergred.  So, we will be watching for copy volume \u003c-\u003e image.","accounts_in_message":[],"_revision_number":19},{"id":"9caec8ef0ede970cdd5ffe01f346b6847557da28","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2018-12-17 18:44:08.000000000","message":"Patch Set 19:\n\nAlso, I think a dependency needs to be added on the os-brick change.","accounts_in_message":[],"_revision_number":19},{"id":"76a7e7b4209c7c1bf486dac5036f2a4090dabc42","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2018-12-18 01:39:59.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/621465/19/Dell_SC_FC/557 : SUCCESS in 2h 52m 22s\n- Dell_SC_iSCSI http://publiclogs.emc.com/621465/19/Dell_SC_iSCSI/544 : SUCCESS in 3h 09m 52s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":19},{"id":"ee1697616730baa40663f8eef49a3b7aedf0c1af","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-18 06:43:44.000000000","message":"Uploaded patch set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"ac84d78ac31ed90595f09a1aa30ff79628e0e732","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-18 06:49:03.000000000","message":"Patch Set 20:\n\n@Jay Bryant, thanks for help addressing gaps for this driver.\n\nDependency is added.\n\nCI is on the way.\n\nAbout copy volume \u003c-\u003e image, we are verifying whether the base-class implementation is sufficient.\n\nUT: Will finish this week.","accounts_in_message":[],"_revision_number":20},{"id":"3ea6e94a85df124ed771b7eda3a6b084d391f435","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-18 07:10:36.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/20/iscsi/ : FAILURE in 15m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/20/smb/ : FAILURE in 15m 18s","accounts_in_message":[],"_revision_number":20},{"id":"7d57f2c1b2eb73ec6306921554d0813df6d144e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-18 07:19:54.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/20/check-not-vote/ext-cinder-zuul/528e1b0 : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":20},{"id":"518250f45bfed9af31110ef458b4416c08bfa461","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-18 07:36:52.000000000","message":"Patch Set 20:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/20/check/fujitsu-eternus-dx-iscsi/599e59f : FAILURE in 47m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/20/check/fujitsu-eternus-dx-fc/d93a076 : FAILURE in 48m 27s","accounts_in_message":[],"_revision_number":20},{"id":"dc190f1234b2db090007a04f9e13e44e25c8c1ae","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-18 07:42:47.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/20/check/dsvm-tempest-kvm/df8701e : SUCCESS in 58m 14s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/20/check/dsvm-tempest-exe/37106a5 : FAILURE in 17m 29s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"485e511f07f535c3db828e418b76995a0df6a773","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-18 07:43:28.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/20/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/8ac5f6c : SUCCESS in 46m 21s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/20/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e25d252 : SUCCESS in 59m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":20},{"id":"548156041f9752acd7efd2071b6afefcc6a65794","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-18 07:50:25.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/20/check/dsvm-tempest-cisco-zonemanager-job/8707230/ : RETRY_LIMIT in 3m 46s","accounts_in_message":[],"_revision_number":20},{"id":"ab4799458af0f3405e021d401905c3555442b980","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-18 08:12:31.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/20/check-cinder/Cinder-ISER-ISCSI/d7fc27d : SUCCESS in 54m 38s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/20/check-cinder/Cinder-ISER-LIO/7d0f432 : SUCCESS in 1h 16m 54s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":20},{"id":"1fd7eb9dda8e977686d6f89ca068a69863acea4d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-18 08:14:18.000000000","message":"Patch Set 20:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/20/check/tempest-dsvm-full-xenial/dec168b/ : SUCCESS in 1h 17m 16s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/20/check/devstack-plugin-ceph-dsvm-xenial/70958a9/ : FAILURE in 1h 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"423013040546afa0e5eae98e350d8e0f1d8d91db","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-18 08:18:39.000000000","message":"Patch Set 20:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/20/check/dsvm-veritas-access-cinder-driver/f7db39f : SUCCESS in 1h 34m 01s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/20/check/dsvm-veritas-access-cinder-iscsi-driver/8940081 : FAILURE in 1h 20m 56s","accounts_in_message":[],"_revision_number":20},{"id":"d892af365c7e21997f18bea034f686f25f873be9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-18 08:48:57.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/20/check/openstack-tox-lower-constraints/6fd973d/ : SUCCESS in 15m 10s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/20/check/openstack-tox-pep8/9043816/ : SUCCESS in 7m 11s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/20/check/openstack-tox-py27/fcb8918/ : SUCCESS in 8m 35s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/20/check/openstack-tox-py35/07b5b65/ : SUCCESS in 8m 22s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/20/check/openstack-tox-py36/21a54a0/ : SUCCESS in 7m 14s\n- openstack-tox-docs http://logs.openstack.org/65/621465/20/check/openstack-tox-docs/2b76572/html/ : SUCCESS in 8m 50s\n- requirements-check http://logs.openstack.org/65/621465/20/check/requirements-check/2afaaba/ : SUCCESS in 3m 22s\n- tempest-full http://logs.openstack.org/65/621465/20/check/tempest-full/8661ff0/ : SUCCESS in 1h 19m 12s\n- neutron-grenade http://logs.openstack.org/65/621465/20/check/neutron-grenade/4936615/ : SUCCESS in 50m 56s\n- grenade-py3 http://logs.openstack.org/65/621465/20/check/grenade-py3/63ba268/ : SUCCESS in 54m 49s\n- tempest-full-py3 http://logs.openstack.org/65/621465/20/check/tempest-full-py3/e374e98/ : SUCCESS in 1h 09m 25s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/20/check/build-openstack-releasenotes/c394f27/html/ : SUCCESS in 12m 03s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/20/check/cinder-tempest-dsvm-lvm-lio-barbican/e01a84a/ : FAILURE in 1h 56m 23s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/20/check/cinder-tox-bandit-baseline/1ff47d7/ : SUCCESS in 7m 02s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/20/check/nova-multiattach/1221702/ : SUCCESS in 58m 29s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/20/check/openstack-tox-functional-py35/3b21f49/ : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/20/check/openstack-tox-functional-py36/6e8c978/ : SUCCESS in 5m 08s\n- cinder-rally-task http://logs.openstack.org/65/621465/20/check/cinder-rally-task/00cc1ea/ : SUCCESS in 38m 22s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/20/check/openstack-tox-pylint/5857627/ : FAILURE in 4m 37s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/20/check/devstack-plugin-ceph-tempest/ac5a7ab/ : SUCCESS in 1h 35m 45s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/20/check/openstack-tox-functional/0c1b8ea/ : SUCCESS in 4m 27s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/20/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/fe7fc05/ : SUCCESS in 1h 00m 32s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/20/check/legacy-tempest-dsvm-lvm-multibackend/a080a39/ : SUCCESS in 1h 11m 55s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/20/check/legacy-tempest-dsvm-full-drbd-devstack/14c5170/ : SUCCESS in 1h 22m 53s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/20/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/024c7fe/ : SUCCESS in 1h 28m 47s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/20/check/tempest-slow/7cdc302/ : SUCCESS in 1h 52m 03s","accounts_in_message":[],"_revision_number":20},{"id":"e1addfa64836dc6b86199c3ea31c5d72e237730e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-18 11:08:21.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/20/check/dsvm-tempest-storpool/f907f39 : SUCCESS in 1h 40m 21s","accounts_in_message":[],"_revision_number":20},{"id":"65715d4714657df569bee84ceb27af0f58d2be11","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-18 12:44:54.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/20/check/EMC_Unity_FC/8884e5d/EMC_Unity_FC/554 : FAILURE in 1h 29m 29s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/20/check/EMC_Unity_iSCSI/792bb46/EMC_Unity_iSCSI/526 : FAILURE in 1h 57m 29s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"63d645ad0f9a7c733a5bc6a7210bd9d7b8649ba5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-18 13:43:23.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/20/check/NEC-ISCSI/30116a6 : SUCCESS in 2h 09m 41s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/20/check/NEC-FC/e68b0d8 : SUCCESS in 2h 13m 02s","accounts_in_message":[],"_revision_number":20},{"id":"5fc58c5708916aa309aba26edfc0770d2ab38c64","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-18 15:02:24.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/20/check/dsvm-tempest-zadara-driver/cd21d90 : SUCCESS in 1h 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0be117071a33941413e06b52b14717044cf0ca7a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-18 16:30:13.000000000","message":"Patch Set 19:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/19/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c9488ed : FAILURE in 2h 21m 32s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/19/check/3par-fc-driver-master-client-pip-ssa02-dsvm/a2998eb : FAILURE in 1h 19m 15s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/19/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/1a3c566 : FAILURE in 1h 42m 10s","accounts_in_message":[],"_revision_number":19},{"id":"ec8ad54eb4f1b750018e363721d6b8ad0253be12","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-18 19:13:56.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/20/check/EMC_XtremIO_FC/9f00db0/EMC_XtremIO_FC/1079 : SUCCESS in 1h 29m 06s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/20/check/EMC_XtremIO_iSCSI/2e48f77/EMC_XtremIO_iSCSI/1101 : SUCCESS in 1h 24m 14s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":20},{"id":"d7382f8869777ac40cf84b5cbe2a95f22790b1e9","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-12-19 00:37:19.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12309 : SUCCESS in 53m 31s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/8729 : SUCCESS in 54m 02s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10512 : FAILURE in 36m 05s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9110 : FAILURE in 35m 14s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10076 : FAILURE in 35m 42s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1157 : SUCCESS in 2h 01m 26s","accounts_in_message":[],"_revision_number":20},{"id":"a1864f62538db6a45f20e014b0d4b93af17bbcf7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-19 08:55:57.000000000","message":"Patch Set 19:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/19/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d6bd1b4 : FAILURE in 43m 01s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/19/check/3par-fc-driver-master-client-pip-ssa02-dsvm/fe25c5a : FAILURE in 39m 36s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/19/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/b4f6339 : FAILURE in 34m 35s","accounts_in_message":[],"_revision_number":19},{"id":"13420cda0acaf3922134969abe521aac50174539","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-19 15:53:31.000000000","message":"Patch Set 19:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/19/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a99ad65 : FAILURE in 1h 30m 38s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/19/check/3par-fc-driver-master-client-pip-ssa02-dsvm/91fa891 : FAILURE in 1h 27m 50s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/19/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/1e85fa7 : SUCCESS in 1h 48m 52s","accounts_in_message":[],"_revision_number":19},{"id":"84c3484dba910b58cbab22e8fc9ea5f9298ab45f","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-20 02:26:16.000000000","message":"Uploaded patch set 21: Commit message was updated.","accounts_in_message":[],"_revision_number":21},{"id":"09a771a2d16ad58946fa6b124a1044dde6063eda","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-20 02:26:54.000000000","message":"Patch Set 21:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/21/check/NEC-ISCSI/8ace631 : FAILURE in 3s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/21/check/NEC-FC/db77903 : FAILURE in 3s","accounts_in_message":[],"_revision_number":21},{"id":"7db412ecaabb848881a5080070e53344189f5ee2","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2018-12-20 02:27:21.000000000","message":"Patch Set 21:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/21/check/dsvm-veritas-access-cinder-driver/0631b6a : FAILURE in 13s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/21/check/dsvm-veritas-access-cinder-iscsi-driver/3c8165d : FAILURE in 13s","accounts_in_message":[],"_revision_number":21},{"id":"a7d089a874880a9533a821a78a092e628afd27ef","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-20 02:27:37.000000000","message":"Patch Set 21:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"d491606c576b72b80db84078f664f7532b7a5517","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2018-12-20 02:33:34.000000000","message":"Patch Set 21:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/444 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"531de778c3980e5c726f46d55de06bb116c88a39","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-20 03:25:17.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/21/iscsi/ : SUCCESS in 58m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/21/smb/ : SUCCESS in 49m 09s","accounts_in_message":[],"_revision_number":21},{"id":"e4fbd107da879b6d73e8095d65d94a1a2f284c8a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-20 03:27:02.000000000","message":"Patch Set 21:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/21/check/fujitsu-eternus-dx-iscsi/c0f039a : FAILURE in 43m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/21/check/fujitsu-eternus-dx-fc/bdc8e03 : FAILURE in 51m 23s","accounts_in_message":[],"_revision_number":21},{"id":"f7cfcedb96367274c343223f977ba870cfe810e4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-20 03:40:19.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/21/check/dsvm-tempest-kvm/fbf325c : SUCCESS in 1h 03m 19s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/21/check/dsvm-tempest-exe/e950a75 : FAILURE in 45m 57s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"993fe7311a2eda803280b6c82dcaa1a60bceb99e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-20 04:13:00.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/21/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1b2f5a7 : SUCCESS in 49m 10s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/21/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/031a66a : FAILURE in 1h 46m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":21},{"id":"fa40b6db930c4b32dbd606cc76099e1f459f2c78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 04:30:44.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/21/check/openstack-tox-lower-constraints/2fc8b37/ : SUCCESS in 9m 35s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/21/check/openstack-tox-pep8/37d0167/ : SUCCESS in 6m 59s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/21/check/openstack-tox-py27/eaf3011/ : FAILURE in 8m 19s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/21/check/openstack-tox-py35/a9857f5/ : SUCCESS in 8m 13s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/21/check/openstack-tox-py36/35d77ed/ : SUCCESS in 8m 01s\n- openstack-tox-docs http://logs.openstack.org/65/621465/21/check/openstack-tox-docs/cfa7467/html/ : SUCCESS in 7m 35s\n- requirements-check http://logs.openstack.org/65/621465/21/check/requirements-check/e687b3b/ : SUCCESS in 3m 02s\n- tempest-full http://logs.openstack.org/65/621465/21/check/tempest-full/fcfac89/ : SUCCESS in 1h 27m 39s\n- neutron-grenade http://logs.openstack.org/65/621465/21/check/neutron-grenade/5ee9523/ : SUCCESS in 50m 10s\n- grenade-py3 http://logs.openstack.org/65/621465/21/check/grenade-py3/a17fff1/ : SUCCESS in 54m 25s\n- tempest-full-py3 http://logs.openstack.org/65/621465/21/check/tempest-full-py3/f335a59/ : SUCCESS in 1h 08m 25s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/21/check/build-openstack-releasenotes/66d2a00/html/ : SUCCESS in 12m 49s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/21/check/cinder-tempest-dsvm-lvm-lio-barbican/0a80863/ : SUCCESS in 2h 00m 20s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/21/check/cinder-tox-bandit-baseline/a8333d9/ : SUCCESS in 6m 23s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/21/check/nova-multiattach/78333c6/ : SUCCESS in 57m 41s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/21/check/openstack-tox-functional-py35/2b9443e/ : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/21/check/openstack-tox-functional-py36/3b97132/ : SUCCESS in 4m 56s\n- cinder-rally-task http://logs.openstack.org/65/621465/21/check/cinder-rally-task/2fc8edb/ : SUCCESS in 33m 26s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/21/check/openstack-tox-pylint/e189e6c/ : FAILURE in 4m 53s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/21/check/devstack-plugin-ceph-tempest/738f6b2/ : SUCCESS in 1h 22m 09s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/21/check/openstack-tox-functional/4111d09/ : SUCCESS in 4m 35s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/21/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/4fa6847/ : SUCCESS in 58m 00s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/21/check/legacy-tempest-dsvm-lvm-multibackend/a14f755/ : SUCCESS in 1h 08m 53s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/21/check/legacy-tempest-dsvm-full-drbd-devstack/ddbab3b/ : SUCCESS in 1h 08m 01s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/21/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/c0adb00/ : SUCCESS in 1h 17m 00s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/21/check/tempest-slow/33d2718/ : SUCCESS in 1h 39m 50s","accounts_in_message":[],"_revision_number":21},{"id":"4fac302dc472622790c27e2755725aacfa7ce4f8","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-20 05:10:30.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"95b43be6ebe52847ee06a139d9b81c7e08ca88f7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-20 05:29:56.000000000","message":"Patch Set 21:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/21/check/tempest-dsvm-full-xenial/88f6863/ : FAILURE in 3h 03m 13s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/21/check/devstack-plugin-ceph-dsvm-xenial/4f2c0ba/ : FAILURE in 3h 02m 00s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"fb8a3b6f53196c337bf95db62e8703517e9ed818","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-20 06:07:26.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/21/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/fad1c41 : SUCCESS in 47m 37s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/21/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/f3b2734 : SUCCESS in 56m 38s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":21},{"id":"0b1dbedd68fbf30e15900d51780ef35d73cff577","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 07:03:27.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/21/check/openstack-tox-lower-constraints/05c097e/ : SUCCESS in 10m 27s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/21/check/openstack-tox-pep8/086b50d/ : SUCCESS in 6m 56s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/21/check/openstack-tox-py27/2677224/ : SUCCESS in 8m 33s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/21/check/openstack-tox-py35/6174c31/ : SUCCESS in 7m 52s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/21/check/openstack-tox-py36/4da167a/ : SUCCESS in 7m 42s\n- openstack-tox-docs http://logs.openstack.org/65/621465/21/check/openstack-tox-docs/694b3a7/html/ : SUCCESS in 7m 11s\n- requirements-check http://logs.openstack.org/65/621465/21/check/requirements-check/3ab8d4f/ : SUCCESS in 2m 16s\n- tempest-full http://logs.openstack.org/65/621465/21/check/tempest-full/b0abbd7/ : SUCCESS in 1h 29m 36s\n- neutron-grenade http://logs.openstack.org/65/621465/21/check/neutron-grenade/a2d7254/ : SUCCESS in 1h 02m 17s\n- grenade-py3 http://logs.openstack.org/65/621465/21/check/grenade-py3/5ad2f33/ : SUCCESS in 49m 33s\n- tempest-full-py3 http://logs.openstack.org/65/621465/21/check/tempest-full-py3/79055d5/ : SUCCESS in 1h 33m 35s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/21/check/build-openstack-releasenotes/eadba99/html/ : SUCCESS in 15m 46s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/21/check/cinder-tempest-dsvm-lvm-lio-barbican/1dd5439/ : SUCCESS in 1h 29m 29s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/21/check/cinder-tox-bandit-baseline/0eee7e0/ : SUCCESS in 6m 25s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/21/check/nova-multiattach/3449a46/ : SUCCESS in 1h 01m 30s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/21/check/openstack-tox-functional-py35/73465e6/ : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/21/check/openstack-tox-functional-py36/65dbf55/ : SUCCESS in 4m 20s\n- cinder-rally-task http://logs.openstack.org/65/621465/21/check/cinder-rally-task/28da7e4/ : SUCCESS in 37m 11s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/21/check/openstack-tox-pylint/f9d646e/ : FAILURE in 4m 33s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/21/check/devstack-plugin-ceph-tempest/3ae32dc/ : SUCCESS in 1h 39m 54s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/21/check/openstack-tox-functional/76c7704/ : SUCCESS in 4m 21s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/21/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/b68da82/ : SUCCESS in 1h 08m 10s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/21/check/legacy-tempest-dsvm-lvm-multibackend/ef28fd8/ : SUCCESS in 1h 26m 30s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/21/check/legacy-tempest-dsvm-full-drbd-devstack/46853be/ : SUCCESS in 1h 15m 00s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/21/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/f255298/ : SUCCESS in 1h 22m 40s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/21/check/tempest-slow/887497c/ : SUCCESS in 1h 45m 46s","accounts_in_message":[],"_revision_number":21},{"id":"7384a75454f0c929b2f1deb4a2420c6b08c7b55c","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-20 07:05:32.000000000","message":"Patch Set 21:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/21/check/kaminario-dsvm-tempest-full-iscsi/91d68f4 : FAILURE in 1m 27s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/21/check/kaminario-dsvm-tempest-full-FC/a83d77d : FAILURE in 1m 27s","accounts_in_message":[],"_revision_number":21},{"id":"8ee2e21be1f34da4e9a359ce1544ecc2212ae311","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-20 09:55:28.000000000","message":"Patch Set 21:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-21 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":21},{"id":"805fc11ba30d9377ff8e64603581702af48ef308","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-20 10:01:29.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/21/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/31dc426 : SUCCESS in 2h 36m 07s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/21/check/3par-fc-driver-master-client-pip-ssa02-dsvm/36db441 : SUCCESS in 1h 53m 10s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/21/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/2b21145 : SUCCESS in 1h 55m 52s","accounts_in_message":[],"_revision_number":21},{"id":"aab064e2697e47c42837952851899ba12eeb7705","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-20 14:13:10.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.57.58.96/logs/65/621465/21/upstream-solidfire-check/cinder-solidfire-iSCSI/909b67f/ : SUCCESS in 35m 05s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":21},{"id":"d410345ebce48dfef058e93ce9c2ccb8610f025d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-20 14:32:42.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/21/iscsi/ : SUCCESS in 9h 21m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/21/smb/ : SUCCESS in 8h 51m 16s","accounts_in_message":[],"_revision_number":21},{"id":"298fa16e9de6dd26e3fb4aaf27990335e1ce516c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-20 14:57:37.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/21/check/dsvm-tempest-storpool/7f355c4 : SUCCESS in 1h 42m 33s","accounts_in_message":[],"_revision_number":21},{"id":"5a143d66525d3e797862451c8d55f39b42cc00d3","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-12-20 17:20:45.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.57.58.96/logs/65/621465/21/upstream-eseries-check/cinder-eseries-iSCSI/5c88940/ : FAILURE in 1h 56m 45s (non-voting)\n- cinder-eseries-FCP http://13.57.58.96/logs/65/621465/21/upstream-eseries-check/cinder-eseries-FCP/03d861c/ : FAILURE in 44m 01s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":21},{"id":"0e3dd96122c8084406bcf59d140f0f86c0c18ab6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-20 17:46:17.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/21/check/EMC_Unity_FC/437290f/EMC_Unity_FC/587 : FAILURE in 1h 44m 57s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/21/check/EMC_Unity_iSCSI/fef537b/EMC_Unity_iSCSI/556 : FAILURE in 1h 57m 23s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"2ed79d9545777c3e7a2813c0d7e700938beb55f1","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-20 22:57:03.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/21/check/synology-dsm-cinder-iscsi/7de18a8 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":21},{"id":"2af314db731ef322562ffc2b6009fffeab98dee6","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-21 01:18:57.000000000","message":"Patch Set 21:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/21/HPMSA-FC : SUCCESS in 37:06.72\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/21/HPMSA-iSCSI : SUCCESS in 35:49.83","accounts_in_message":[],"_revision_number":21},{"id":"e5f710783d59bd6459831408a9b4f9ad6d5070c2","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-21 01:58:56.000000000","message":"Patch Set 21:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/21/Lenovo-FC : SUCCESS in 36:34.34\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/21/Lenovo-iSCSI : SUCCESS in 36:03.37","accounts_in_message":[],"_revision_number":21},{"id":"4c976c11cb044cb5a125123205972f01d8f93c47","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-21 07:49:34.000000000","message":"Patch Set 21:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/21/check/dsvm-tempest-zadara-driver/25c7cd7 : SUCCESS in 1h 23m 14s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"2bd44f95ee6ad94fa44628fe8048c48c8b44bab1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-21 10:54:52.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.57.58.96/logs/65/621465/21/upstream-check/cinder-cDOT-iSCSI/42c17b5/ : SUCCESS in 28m 38s\n- cinder-cDOT-NFS http://13.57.58.96/logs/65/621465/21/upstream-check/cinder-cDOT-NFS/469884f/ : POST_FAILURE in 17m 13s\n- cinder-cDOT-FCP http://13.57.58.96/logs/65/621465/21/upstream-check/cinder-cDOT-FCP/e18e609/ : SUCCESS in 48m 57s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":21},{"id":"fd03e3b689ede3ad115cf4375efecfb728672fc9","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-12-22 01:13:35.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12366 : SUCCESS in 55m 08s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/8793 : SUCCESS in 57m 38s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10568 : FAILURE in 38m 59s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9174 : FAILURE in 39m 49s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10139 : FAILURE in 37m 20s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1192 : SUCCESS in 1h 59m 01s","accounts_in_message":[],"_revision_number":21},{"id":"86d516420610ec4011e5817bc827864d1f8a869d","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-24 08:12:13.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"d0e2fdc9c1d0b39648f43d1133505131ae6b7bea","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-12-24 08:27:58.000000000","message":"Patch Set 22:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":22},{"id":"e48e1980bfc4cf46953466470d9b9d82f2520ccc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-24 09:02:58.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/22/iscsi/ : SUCCESS in 50m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/22/smb/ : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":22},{"id":"e4e12225a580a3d4f9077ba907cb2ed4e13f995c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-24 09:21:50.000000000","message":"Patch Set 22:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/22/check/tempest-dsvm-full-xenial/735d6ca/ : SUCCESS in 1h 08m 09s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/22/check/devstack-plugin-ceph-dsvm-xenial/2f75e77/ : SUCCESS in 1h 09m 16s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"3336315500cc3930b5f151d40ea5e41827e5ac32","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-24 09:42:08.000000000","message":"Patch Set 22:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/22/check/fujitsu-eternus-dx-iscsi/1d95184 : SUCCESS in 1h 25m 14s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/22/check/fujitsu-eternus-dx-fc/c698fca : SUCCESS in 1h 28m 21s","accounts_in_message":[],"_revision_number":22},{"id":"cdf73b05334483ad6bbfdb255d3f2cb4f0010eee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-24 10:00:13.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/22/check/openstack-tox-lower-constraints/141bd37/ : SUCCESS in 9m 45s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/22/check/openstack-tox-pep8/d3e6eac/ : SUCCESS in 11m 45s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/22/check/openstack-tox-py27/e7c6e2d/ : SUCCESS in 10m 24s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/22/check/openstack-tox-py35/2c7a4d6/ : FAILURE in 13m 02s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/22/check/openstack-tox-py36/0db7a15/ : SUCCESS in 8m 28s\n- openstack-tox-docs http://logs.openstack.org/65/621465/22/check/openstack-tox-docs/aae106a/html/ : SUCCESS in 7m 10s\n- requirements-check http://logs.openstack.org/65/621465/22/check/requirements-check/c0340e5/ : SUCCESS in 2m 44s\n- tempest-full http://logs.openstack.org/65/621465/22/check/tempest-full/c041459/ : SUCCESS in 1h 27m 43s\n- neutron-grenade http://logs.openstack.org/65/621465/22/check/neutron-grenade/0d71d5c/ : SUCCESS in 47m 19s\n- grenade-py3 http://logs.openstack.org/65/621465/22/check/grenade-py3/74af753/ : SUCCESS in 1h 06m 31s\n- tempest-full-py3 http://logs.openstack.org/65/621465/22/check/tempest-full-py3/1360ded/ : SUCCESS in 1h 03m 34s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/22/check/build-openstack-releasenotes/7d7a1fe/html/ : SUCCESS in 12m 17s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/22/check/cinder-tempest-dsvm-lvm-lio-barbican/bc8a041/ : SUCCESS in 1h 35m 08s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/22/check/cinder-tox-bandit-baseline/2ca5f10/ : SUCCESS in 6m 34s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/22/check/nova-multiattach/b1f7573/ : SUCCESS in 1h 04m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/22/check/openstack-tox-functional-py35/4c12237/ : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/22/check/openstack-tox-functional-py36/a489aa5/ : SUCCESS in 5m 16s\n- cinder-rally-task http://logs.openstack.org/65/621465/22/check/cinder-rally-task/0879763/ : SUCCESS in 50m 48s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/22/check/openstack-tox-pylint/29b4361/ : FAILURE in 4m 28s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/22/check/devstack-plugin-ceph-tempest/441a804/ : FAILURE in 1h 36m 38s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/22/check/openstack-tox-functional/174933d/ : SUCCESS in 4m 36s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/22/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/32881a7/ : SUCCESS in 1h 16m 58s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/22/check/legacy-tempest-dsvm-lvm-multibackend/896f0a3/ : SUCCESS in 1h 14m 46s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/22/check/legacy-tempest-dsvm-full-drbd-devstack/ee57510/ : SUCCESS in 1h 15m 30s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/22/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/ba9fb4e/ : SUCCESS in 1h 22m 22s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/22/check/tempest-slow/63eb157/ : SUCCESS in 1h 41m 09s","accounts_in_message":[],"_revision_number":22},{"id":"c02fd219510ca8e73ec2a5c7488e094d4ffd11cb","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-24 10:04:16.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/22/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/8f7acf9 : SUCCESS in 44m 03s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/22/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/8359f32 : SUCCESS in 1h 10m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":22},{"id":"c1c84a574e5c51899c5e4bf1315420e095c7838e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-24 10:04:38.000000000","message":"Patch Set 22:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"49bb32877012a5e34bd9f7bee99e6437b324a36f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-24 10:05:05.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/22/check/dsvm-tempest-kvm/411875e : SUCCESS in 1h 19m 25s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/22/check/dsvm-tempest-exe/6d058b5 : SUCCESS in 1h 24m 27s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"4e957a77a4eb4d40d22dab3579ad5985b1ea161e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-24 10:07:09.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/22/check-not-vote/ext-cinder-zuul/4cfd082 : FAILURE in 28m 23s","accounts_in_message":[],"_revision_number":22},{"id":"e4f2dc82486da9a18f90403fafa4002ddd4636b1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-24 10:15:33.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/22/check/dsvm-tempest-cisco-zonemanager-job/1304c9b/ : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":22},{"id":"7c1d2fc638f875bf4c0eb05cee24971b84a94a64","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-24 11:55:13.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/22/check/EMC_Unity_FC/04d5e88/EMC_Unity_FC/612 : FAILURE in 1h 18m 53s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/22/check/EMC_Unity_iSCSI/31745fd/EMC_Unity_iSCSI/583 : FAILURE in 1h 05m 06s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"4308c836c85b86765fbda060f206214d19a204ea","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-24 13:15:06.000000000","message":"Patch Set 22:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/22/check/NEC-ISCSI/8529679 : FAILURE in 55m 13s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/22/check/NEC-FC/40dcf56 : FAILURE in 1h 00m 02s","accounts_in_message":[],"_revision_number":22},{"id":"52c254445e2c2f83f49d2bda4042501c7daa8f0c","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-24 13:54:37.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"a1e469625e9b01351ec407b68fbc348467d1e840","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-24 14:22:10.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/22/check-not-vote/ext-cinder-zuul/ee21cff : FAILURE in 26m 56s","accounts_in_message":[],"_revision_number":22},{"id":"e570b1b1b06daa1c78d4df576a7365e8bd3c456f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-24 14:43:02.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/22/iscsi/ : FAILURE in 36m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/22/smb/ : FAILURE in 47m 55s","accounts_in_message":[],"_revision_number":22},{"id":"8c908c9e0ffa3483b1a4dcc30daaa70936e72eec","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-24 14:52:01.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/22/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/40451df : SUCCESS in 46m 28s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/22/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/8f95f40 : SUCCESS in 57m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":22},{"id":"fd4d809d12c914902aa4215ef1e001453112fb5b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-24 15:08:44.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/22/check/dsvm-tempest-cisco-zonemanager-job/08b9498/ : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":22},{"id":"3f51b14bab90783a42441dcc24f7b0516d6fbf4a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-24 15:15:01.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/22/check/EMC_XtremIO_FC/75b4f15/EMC_XtremIO_FC/1137 : SUCCESS in 1h 27m 13s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/22/check/EMC_XtremIO_iSCSI/77830c9/EMC_XtremIO_iSCSI/1165 : SUCCESS in 1h 20m 49s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":22},{"id":"065ec9545ec46a6f6e4ce62d535e8f37ad6e9c99","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-24 15:25:13.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/22/check/dsvm-tempest-kvm/568c5f3 : FAILURE in 1h 28m 26s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/22/check/dsvm-tempest-exe/a8f5690 : SUCCESS in 1h 11m 42s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"1fc5faaba06666dd4eb9952a88142edeed5e1b07","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-24 15:26:59.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/22/check/dsvm-tempest-storpool/eec89fe : SUCCESS in 1h 44m 13s","accounts_in_message":[],"_revision_number":22},{"id":"361dcc2851b16548a2998ab8ba00f98f07d233a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-24 15:54:28.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/22/check/openstack-tox-lower-constraints/f7f56a9/ : SUCCESS in 9m 12s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/22/check/openstack-tox-pep8/0c0119a/ : SUCCESS in 6m 54s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/22/check/openstack-tox-py27/c007af0/ : SUCCESS in 9m 04s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/22/check/openstack-tox-py35/18a15e6/ : SUCCESS in 8m 06s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/22/check/openstack-tox-py36/108ba6f/ : SUCCESS in 7m 34s\n- openstack-tox-docs http://logs.openstack.org/65/621465/22/check/openstack-tox-docs/e7bd907/html/ : SUCCESS in 8m 16s\n- requirements-check http://logs.openstack.org/65/621465/22/check/requirements-check/d3bfb69/ : SUCCESS in 2m 28s\n- tempest-full http://logs.openstack.org/65/621465/22/check/tempest-full/9a125e0/ : SUCCESS in 1h 21m 33s\n- neutron-grenade http://logs.openstack.org/65/621465/22/check/neutron-grenade/1c8bc1f/ : SUCCESS in 59m 46s\n- grenade-py3 http://logs.openstack.org/65/621465/22/check/grenade-py3/abad515/ : SUCCESS in 56m 33s\n- tempest-full-py3 http://logs.openstack.org/65/621465/22/check/tempest-full-py3/8912bcd/ : SUCCESS in 1h 11m 20s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/22/check/build-openstack-releasenotes/bad55dd/html/ : SUCCESS in 14m 45s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/22/check/cinder-tempest-dsvm-lvm-lio-barbican/69ce32d/ : SUCCESS in 1h 56m 46s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/22/check/cinder-tox-bandit-baseline/18a0fa3/ : SUCCESS in 6m 37s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/22/check/nova-multiattach/18d14e7/ : SUCCESS in 1h 01m 52s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/22/check/openstack-tox-functional-py35/783ffcd/ : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/22/check/openstack-tox-functional-py36/1f33afb/ : SUCCESS in 4m 15s\n- cinder-rally-task http://logs.openstack.org/65/621465/22/check/cinder-rally-task/a3c5894/ : SUCCESS in 40m 53s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/22/check/openstack-tox-pylint/f6cc93f/ : FAILURE in 4m 09s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/22/check/devstack-plugin-ceph-tempest/523d2cc/ : SUCCESS in 1h 38m 05s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/22/check/openstack-tox-functional/9ecc1b0/ : SUCCESS in 5m 03s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/22/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/02fda0b/ : SUCCESS in 1h 06m 35s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/22/check/legacy-tempest-dsvm-lvm-multibackend/6d6bae8/ : SUCCESS in 1h 16m 56s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/22/check/legacy-tempest-dsvm-full-drbd-devstack/d5ee216/ : SUCCESS in 1h 25m 04s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/22/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/207f76c/ : SUCCESS in 1h 26m 08s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/22/check/tempest-slow/5cd5028/ : SUCCESS in 1h 48m 46s","accounts_in_message":[],"_revision_number":22},{"id":"ee74ec03284e6b8906c74e6e0b76488953d1dee9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-24 16:02:42.000000000","message":"Patch Set 22:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/22/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f3d8bbb : FAILURE in 5m 28s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/22/check/3par-fc-driver-master-client-pip-ssa02-dsvm/f5270bf : FAILURE in 2m 23s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/22/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/02b6e4c : FAILURE in 23s","accounts_in_message":[],"_revision_number":22},{"id":"ec928cd6c229aa6d2ac9dee1b194f67ad1e865c8","author":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"date":"2018-12-24 16:34:52.000000000","message":"Patch Set 22:\n\n(2 comments)\n\nSome small nits.","accounts_in_message":[],"_revision_number":22},{"id":"d85613b1a20298454fb615e75629ae907af00267","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-24 16:46:02.000000000","message":"Patch Set 22:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-22 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":22},{"id":"6f1e84f339cea47f548f3adb5497a9bb3bb8bec1","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-24 16:48:00.000000000","message":"Patch Set 22:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/22/check/synology-dsm-cinder-iscsi/2c19955 : FAILURE in 52m 20s","accounts_in_message":[],"_revision_number":22},{"id":"eaa638fa5d002b7848fb6853bec177632390e070","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-24 16:48:51.000000000","message":"Patch Set 22:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-22/nedge_iscsi : FAILURE in 51m 40s","accounts_in_message":[],"_revision_number":22},{"id":"427782cd531854fcdd017c668fd7a307397031ab","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-24 16:56:58.000000000","message":"Patch Set 22:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/22/check/tempest-dsvm-full-xenial/0dc8528/ : FAILURE in 3h 01m 59s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/22/check/devstack-plugin-ceph-dsvm-xenial/d8f2970/ : SUCCESS in 57m 55s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"f1a4fe0a9fd4e009cd7c54576011530e3c08f6cb","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-24 17:09:18.000000000","message":"Patch Set 22:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-22-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"9c0478aef81f75db8fcb0ceed188faba9e34d6e9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-24 19:19:55.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/22/check/dsvm-tempest-zadara-driver/540d9be : FAILURE in 1h 51m 52s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d795aa5f254a49d5c5fe5e6be3f2a0d25acdd9a6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-24 19:58:02.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/22/check/EMC_Unity_FC/f0094af/EMC_Unity_FC/618 : FAILURE in 1h 18m 57s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/22/check/EMC_Unity_iSCSI/c4daf80/EMC_Unity_iSCSI/589 : FAILURE in 1h 04m 56s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"202ce9add15db9e4890f153dd34f971ad1b2c894","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-24 21:00:21.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/65/621465/22/check/EMC_VxFlexOS/c0082e8/EMC_VxFlexOS/704 : SUCCESS in 1h 52m 21s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":22},{"id":"5aa5fba670aab579440ddfd92ed045b2f8f42592","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-24 21:23:14.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.57.58.96/logs/65/621465/22/upstream-check/cinder-cDOT-iSCSI/7551b1d/ : POST_FAILURE in 1h 36m 58s\n- cinder-cDOT-NFS http://13.57.58.96/logs/65/621465/22/upstream-check/cinder-cDOT-NFS/0b91b8c/ : POST_FAILURE in 1h 53m 18s\n- cinder-cDOT-FCP http://13.57.58.96/logs/65/621465/22/upstream-check/cinder-cDOT-FCP/c9a5929/ : POST_FAILURE in 43m 51s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":22},{"id":"a45edb0f6a235af1ddb2f5cc2f18f0c502fb4cc9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-24 21:55:12.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/22/check/EMC_XtremIO_FC/fe14d2c/EMC_XtremIO_FC/1142 : SUCCESS in 1h 29m 43s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/22/check/EMC_XtremIO_iSCSI/72d38b2/EMC_XtremIO_iSCSI/1170 : SUCCESS in 1h 20m 14s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":22},{"id":"8c466217806761816a77baf6eb17405162caa5ea","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-24 22:03:40.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.57.58.96/logs/65/621465/22/upstream-solidfire-check/cinder-solidfire-iSCSI/64be49e/ : POST_FAILURE in 57m 01s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":22},{"id":"f906e46eb173a8c087660e12cb5725678907d7b6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2018-12-25 03:35:04.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/621465/22/Dell_SC_FC/576 : SUCCESS in 2h 53m 24s\n- Dell_SC_iSCSI http://publiclogs.emc.com/621465/22/Dell_SC_iSCSI/560 : SUCCESS in 3h 06m 19s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":22},{"id":"624292e14c0e1d1e1eedd2bd774981fb42d2c0e8","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-25 08:16:18.000000000","message":"Patch Set 22:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/22/Lenovo-FC : FAILURE in 1:19:07\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/22/Lenovo-iSCSI : FAILURE in 1:14:58","accounts_in_message":[],"_revision_number":22},{"id":"8f8047e89fdfdc81d185d21bb0400c16c2cee6b2","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-25 10:26:56.000000000","message":"Patch Set 22:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/22/HPMSA-FC : SUCCESS in 37:40.40\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/22/HPMSA-iSCSI : SUCCESS in 37:08.68","accounts_in_message":[],"_revision_number":22},{"id":"e83f37fccaa395a55b637ba8f0ec5742af1f6d6b","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2018-12-25 21:56:09.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.57.58.96/logs/65/621465/22/upstream-eseries-check/cinder-eseries-iSCSI/f58dd50/ : POST_FAILURE in 11m 00s (non-voting)\n- cinder-eseries-FCP http://13.57.58.96/logs/65/621465/22/upstream-eseries-check/cinder-eseries-FCP/c96fd76/ : POST_FAILURE in 37m 04s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":22},{"id":"9bda1508dfba3ea9221497608f3b9caa5c7eb167","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-26 08:09:44.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"e8bf1b6fd403ec0164496117119407d33e2fcaf1","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2018-12-26 08:15:58.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"0fd54e89df14c3a91aaeb1edeacf2606feec49b0","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2018-12-26 08:31:15.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/495 : ABORTED\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/712 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/880 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"08dfe005439297eb50bec3fb3feb3e1109a89176","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-26 09:10:28.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/23/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/ae5b5b8 : SUCCESS in 42m 38s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/23/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/3c57b31 : SUCCESS in 1h 00m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":23},{"id":"7cd7ec3a4fe57231993eb8bd53183e7173e36b5c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-26 09:12:44.000000000","message":"Patch Set 23:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/23/check/tempest-dsvm-full-xenial/95898b8/ : SUCCESS in 1h 02m 35s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/23/check/devstack-plugin-ceph-dsvm-xenial/9942ad1/ : FAILURE in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"7b27ae36aea0680ed4b35cf3bab531562322ff0a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2018-12-26 09:16:25.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/23/iscsi/ : SUCCESS in 51m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/23/smb/ : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":23},{"id":"6d9e98ec2c294a4cc13ebb3da32b06b83f993a4b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2018-12-26 09:26:52.000000000","message":"Patch Set 23:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/23 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"df1c3e618f0c7c205c77bc569775bca378ddd0de","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2018-12-26 09:28:48.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/23/check/dsvm-tempest-cisco-zonemanager-job/39f7a3e/ : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":23},{"id":"686cf0d58c06bbaf2309d8547568ea165a3355ef","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2018-12-26 09:32:05.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/23/check/dsvm-tempest-kvm/ca3b74f : SUCCESS in 1h 16m 52s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/23/check/dsvm-tempest-exe/0661ede : FAILURE in 1h 19m 39s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"6a3bb8899b3b4066020922617b23d280a1255551","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-26 10:21:59.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/23/check/openstack-tox-lower-constraints/c9aa349/ : SUCCESS in 10m 30s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/23/check/openstack-tox-pep8/93012cb/ : SUCCESS in 7m 05s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/23/check/openstack-tox-py27/060b16c/ : SUCCESS in 8m 02s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/23/check/openstack-tox-py35/48ea258/ : SUCCESS in 9m 58s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/23/check/openstack-tox-py36/a28851a/ : SUCCESS in 7m 16s\n- openstack-tox-docs http://logs.openstack.org/65/621465/23/check/openstack-tox-docs/1100ec2/html/ : SUCCESS in 7m 36s\n- requirements-check http://logs.openstack.org/65/621465/23/check/requirements-check/eba3e56/ : SUCCESS in 2m 24s\n- tempest-full http://logs.openstack.org/65/621465/23/check/tempest-full/d58771b/ : SUCCESS in 1h 10m 16s\n- neutron-grenade http://logs.openstack.org/65/621465/23/check/neutron-grenade/218189a/ : SUCCESS in 54m 01s\n- grenade-py3 http://logs.openstack.org/65/621465/23/check/grenade-py3/02fc5fc/ : SUCCESS in 1h 02m 14s\n- tempest-full-py3 http://logs.openstack.org/65/621465/23/check/tempest-full-py3/d56934b/ : SUCCESS in 1h 01m 07s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/23/check/build-openstack-releasenotes/0ce9a7d/html/ : SUCCESS in 13m 18s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/23/check/cinder-tempest-dsvm-lvm-lio-barbican/7ddaef6/ : SUCCESS in 2h 04m 11s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/23/check/cinder-tox-bandit-baseline/2dcc6f8/ : SUCCESS in 6m 15s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/23/check/nova-multiattach/273bf3b/ : SUCCESS in 1h 00m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/23/check/openstack-tox-functional-py35/a1eb205/ : SUCCESS in 4m 33s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/23/check/openstack-tox-functional-py36/2ad643e/ : SUCCESS in 4m 13s\n- cinder-rally-task http://logs.openstack.org/65/621465/23/check/cinder-rally-task/5282b5a/ : SUCCESS in 37m 49s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/23/check/openstack-tox-pylint/1c0dd33/ : FAILURE in 4m 16s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/23/check/devstack-plugin-ceph-tempest/b132105/ : SUCCESS in 1h 31m 29s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/23/check/openstack-tox-functional/571d492/ : SUCCESS in 4m 30s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/23/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/9f765bf/ : SUCCESS in 51m 26s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/23/check/legacy-tempest-dsvm-lvm-multibackend/209d812/ : SUCCESS in 1h 24m 28s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/23/check/legacy-tempest-dsvm-full-drbd-devstack/a785adc/ : SUCCESS in 1h 11m 38s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/23/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/0b86cef/ : SUCCESS in 1h 20m 01s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/23/check/tempest-slow/4ffc23d/ : SUCCESS in 1h 47m 37s","accounts_in_message":[],"_revision_number":23},{"id":"ae695e2ef195d496da75765c2485b961903eb56e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2018-12-26 10:40:46.000000000","message":"Patch Set 23:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/23/check/fujitsu-eternus-dx-iscsi/a4a78db : SUCCESS in 2h 25m 48s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/23/check/fujitsu-eternus-dx-fc/a3566f9 : SUCCESS in 2h 28m 11s","accounts_in_message":[],"_revision_number":23},{"id":"655230c7ee4ce91ad801a68ea19d3910360ef591","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2018-12-26 10:50:24.000000000","message":"Patch Set 23:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-23/nedge_iscsi : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":23},{"id":"67aa2c6609b3fbcd6276a5e277474bcc466c3187","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2018-12-26 10:54:28.000000000","message":"Patch Set 23:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/23/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1801c51 : FAILURE in 5m 22s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/23/check/3par-fc-driver-master-client-pip-ssa02-dsvm/63732da : FAILURE in 2m 22s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/23/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/4d92717 : FAILURE in 23s","accounts_in_message":[],"_revision_number":23},{"id":"55eb149a41cd297666186ab897084e1c88567cff","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-12-26 11:08:52.000000000","message":"Patch Set 23:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-23 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":23},{"id":"494a7b8d260913f886fb019fb0747b52c6574bca","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2018-12-26 11:12:13.000000000","message":"Patch Set 23:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/23/check/Cinder-AS13000/2fdfe7c : SUCCESS in 2h 22m 44s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/23/check/cinder-test-iscsi/2608f01 : FAILURE in 30m 56s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/23/check/cinder-test-fc/5cec549 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":23},{"id":"4370e9c52211c87d449eb0326fae7c6f11e38a0b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-12-26 11:13:42.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/23/check-cinder/Cinder-ISER-ISCSI/7e12109 : SUCCESS in 58m 07s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/23/check-cinder/Cinder-ISER-LIO/b7160e6 : SUCCESS in 1h 03m 55s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":23},{"id":"bbb251d7eb32c1e124d555c0eb694d5a0bfd6736","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2018-12-26 11:14:08.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/23/check/synology-dsm-cinder-iscsi/450b2ca : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":23},{"id":"a13248e0aa0c46b954e63315e1e13059c2c1f488","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2018-12-26 11:35:41.000000000","message":"Patch Set 23:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-23-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"653b5f7f7370ec6e03291e35f4b6eea14d5d49f7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2018-12-26 11:50:22.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/23/check/EMC_Unity_FC/a3dddb2/EMC_Unity_FC/641 : FAILURE in 1h 21m 49s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/23/check/EMC_Unity_iSCSI/829eff4/EMC_Unity_iSCSI/612 : FAILURE in 1h 03m 24s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"0ca90328fdcfdee0fdba9565a8ab06fc86a6767c","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2018-12-26 12:08:52.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/23/check/kaminario-dsvm-tempest-full-iscsi/b4a4462 : SUCCESS in 1h 46m 15s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/23/check/kaminario-dsvm-tempest-full-FC/e81bed3 : SUCCESS in 1h 40m 01s","accounts_in_message":[],"_revision_number":23},{"id":"a6668dea3828c6730869009fa43d4238601e7955","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2018-12-26 12:47:23.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/23/check/dsvm-tempest-storpool/841ab36 : SUCCESS in 1h 45m 37s","accounts_in_message":[],"_revision_number":23},{"id":"0b12baad267c3532c68e24265686f996daaf7f3c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-26 12:47:55.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/23/check/EMC_XtremIO_FC/7f2c9e7/EMC_XtremIO_FC/1165 : SUCCESS in 1h 28m 21s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/23/check/EMC_XtremIO_iSCSI/11e6258/EMC_XtremIO_iSCSI/1192 : SUCCESS in 1h 30m 53s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":23},{"id":"88dd14b685088e032d23c1b6dd0218f6df9413aa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2018-12-26 12:59:38.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.57.58.96/logs/65/621465/23/upstream-check/cinder-cDOT-iSCSI/0e43362/ : POST_FAILURE in 11m 00s\n- cinder-cDOT-NFS http://13.57.58.96/logs/65/621465/23/upstream-check/cinder-cDOT-NFS/722dc6a/ : POST_FAILURE in 1h 34m 23s\n- cinder-cDOT-FCP http://13.57.58.96/logs/65/621465/23/upstream-check/cinder-cDOT-FCP/ad38061/ : POST_FAILURE in 53m 52s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"9ab7f9032494e5d293e1cde132e81c0214f42f14","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2018-12-26 13:42:46.000000000","message":"Patch Set 23:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/23/HPMSA-FC : SUCCESS in 35:14.84\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/23/HPMSA-iSCSI : SUCCESS in 34:32.73","accounts_in_message":[],"_revision_number":23},{"id":"e415ff1bea7d00def01c72254e27c435750782dd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2018-12-26 13:58:34.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.57.58.96/logs/65/621465/23/upstream-solidfire-check/cinder-solidfire-iSCSI/4c0e2b9/ : POST_FAILURE in 11m 00s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":23},{"id":"7fb3b6e44c91614ab5e5dc2cecfe76f1d26d3f92","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-12-26 14:53:35.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/65/621465/23/check/EMC_VxFlexOS/4ddd84d/EMC_VxFlexOS/717 : SUCCESS in 1h 49m 49s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":23},{"id":"d035baf57ad9c11b7acdf253637a9cf09cf17cb1","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2018-12-26 15:09:42.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/23/check/dsvm-tempest-zadara-driver/80cb23a : FAILURE in 1h 52m 38s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"8c20a61e010966b671fe588e54527e39a7111df8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2018-12-26 16:35:43.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/23/check/dsvm-tempest-infinibox-fc/aa8a0e9 : SUCCESS in 1h 03m 35s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/23/check/dsvm-tempest-infinibox-iscsi/3d1fb83 : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":23},{"id":"f035263b225dcffd16d34a2ce29ba25dfd9722e9","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2018-12-26 18:02:10.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/23/check/NEC-ISCSI/16699c9 : SUCCESS in 2h 42m 28s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/23/check/NEC-FC/8030c20 : SUCCESS in 2h 45m 43s","accounts_in_message":[],"_revision_number":23},{"id":"2fe8f403cb50647edce35b775229b23e5667fa0e","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2018-12-26 22:09:21.000000000","message":"Patch Set 23:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/23/Lenovo-FC : FAILURE in 1:09:15\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/23/Lenovo-iSCSI : FAILURE in 1:08:54","accounts_in_message":[],"_revision_number":23},{"id":"46bb7f67549753bcb08947211191ad438dc8b1c4","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2018-12-26 22:45:37.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/621465/23/Dell_SC_FC/588 : SUCCESS in 2h 52m 36s\n- Dell_SC_iSCSI http://publiclogs.emc.com/621465/23/Dell_SC_iSCSI/572 : SUCCESS in 3h 07m 01s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":23},{"id":"f69e7cedc9fdbffada4b108e2903e9f085798360","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2018-12-27 21:13:11.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12486 : FAILURE in 1h 14m 44s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/8894 : FAILURE in 15s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10661 : FAILURE in 35m 53s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9273 : FAILURE in 39m 00s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10239 : FAILURE in 35m 04s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1251 : SUCCESS in 1h 54m 17s","accounts_in_message":[],"_revision_number":23},{"id":"a06288f98e489553310b5f12028ba6b9550f6770","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2018-12-30 06:17:52.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* noop-test http://us.linbit.com:8080/CI-SANDBOX/65/531 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"e6264c45980ff19e265cb6055af69ddc65f4c2db","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2018-12-30 06:30:14.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-SANDBOX/65/532 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"34ff08e7b127a3891fe2f6d4aabd583fd85d2b0a","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2018-12-30 07:01:41.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-SANDBOX/65/533 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"d565a018ff21d1905de13de2ac0bbc3c3772ebb2","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2018-12-30 08:34:12.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-SANDBOX/65/536 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"0ba303a9c5eaa38cea68b4ffb9a2b41e66bbe08f","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2018-12-30 10:11:38.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-SANDBOX/65/537 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"b9242fac72a5ab51f5e350339ef0db9be17bc359","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2018-12-31 04:38:32.000000000","message":"Patch Set 23:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/552 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"fe199ad0861df3a322fc69ea89601e74377fb79c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-02 12:47:53.000000000","message":"Patch Set 23:\n\nBuild Started http://silpixa00399130:8080/job/CI_Environment/30/","accounts_in_message":[],"_revision_number":23},{"id":"77511591cc2f4f3b1287a4886d03381c49393d6d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-02 12:47:56.000000000","message":"Patch Set 23:\n\nBuild Failed \n\nhttp://silpixa00399130:8080/job/CI_Environment/30/ : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"3edc6e0bc6f3dd48f123e04f7f6a60db7824ca94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-02 12:50:18.000000000","message":"Patch Set 22:\n\nBuild Started http://silpixa00399130:8080/job/CI_Environment/58/","accounts_in_message":[],"_revision_number":22},{"id":"17d0bc819f39abbc6ffde96f6c6bd270daf32c84","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-02 12:50:20.000000000","message":"Patch Set 22:\n\nBuild Failed \n\nhttp://silpixa00399130:8080/job/CI_Environment/58/ : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"ae7e01d8e88ea149abeabacc386a5f055c53adb5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-03 16:51:37.000000000","message":"Patch Set 23:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/23/check-not-vote/ext-cinder-zuul/55df471 : SUCCESS in 45m 06s","accounts_in_message":[],"_revision_number":23},{"id":"01d1bbecf28524762b7162384ba2993a1b194139","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-04 05:13:54.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"4b4852fb53d0e7633cc0d889b0fd8f72ce5ae072","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-04 06:07:02.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/24/iscsi/ : SUCCESS in 52m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/24/smb/ : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":24},{"id":"5e0b32a3c2cf7b88a9b9c5fafb4162691e6caa30","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-04 06:14:07.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/24/check-cinder/Cinder-ISER-ISCSI/8df0087 : SUCCESS in 58m 43s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/24/check-cinder/Cinder-ISER-LIO/8b427c5 : SUCCESS in 56m 45s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":24},{"id":"fccf41f35cf3f25f3cd76a0b5a06cd0a34bcde4f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-04 06:17:04.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/24/check/dsvm-tempest-infinibox-fc/3726684 : SUCCESS in 56m 32s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/24/check/dsvm-tempest-infinibox-iscsi/f41e407 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":24},{"id":"3189b86fcfddf26b2f0794b2954b734d51fc6262","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-04 06:21:59.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/24/check/tempest-dsvm-full-xenial/277b072/ : SUCCESS in 1h 07m 43s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/24/check/devstack-plugin-ceph-dsvm-xenial/b86ab82/ : SUCCESS in 57m 44s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"f20918c95bd8ea50d672abc38b2c7af78373d202","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-04 06:23:40.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/24/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/b1634c9 : SUCCESS in 45m 26s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/24/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/6be717e : SUCCESS in 1h 09m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":24},{"id":"c928b7b6fddfaa1dca69c63952ef3050ae8d0951","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 06:29:24.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/24/check/dsvm-tempest-kvm/f8855a3 : SUCCESS in 1h 13m 54s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/24/check/dsvm-tempest-exe/9836b78 : SUCCESS in 58m 51s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"9f3e9db3042682389fea3cdac596d5939ef5fd03","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-04 06:32:44.000000000","message":"Patch Set 24:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/24 : SUCCESS","accounts_in_message":[],"_revision_number":24},{"id":"1b9b490825f3ce7fc55c35776d9a7f643a0ee85c","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-04 06:52:21.000000000","message":"Patch Set 24:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/24/check/dsvm-veritas-access-cinder-driver/58437a3 : FAILURE in 1h 27m 18s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/24/check/dsvm-veritas-access-cinder-iscsi-driver/1e3537d : FAILURE in 1h 37m 26s","accounts_in_message":[],"_revision_number":24},{"id":"ba1d8a1058484f77d8b0b84ccecacb6dfdd436b7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-04 07:12:17.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/24/check/dsvm-tempest-cisco-zonemanager-job/e038269/ : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":24},{"id":"9bfafd67c0dc7fbbbba3695c3c9d7bb2471d7acf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-04 07:15:42.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/24/check/openstack-tox-lower-constraints/bf9980b/ : SUCCESS in 10m 53s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/24/check/openstack-tox-pep8/2927dba/ : SUCCESS in 7m 27s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/24/check/openstack-tox-py27/8f687d2/ : SUCCESS in 7m 59s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/24/check/openstack-tox-py35/8ba182e/ : SUCCESS in 8m 05s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/24/check/openstack-tox-py36/01f5936/ : FAILURE in 7m 40s\n- openstack-tox-docs http://logs.openstack.org/65/621465/24/check/openstack-tox-docs/6ee5b17/html/ : SUCCESS in 7m 52s\n- requirements-check http://logs.openstack.org/65/621465/24/check/requirements-check/76615ad/ : SUCCESS in 2m 42s\n- tempest-full http://logs.openstack.org/65/621465/24/check/tempest-full/5da8e2d/ : SUCCESS in 1h 28m 06s\n- neutron-grenade http://logs.openstack.org/65/621465/24/check/neutron-grenade/c5f3f1d/ : SUCCESS in 1h 00m 53s\n- grenade-py3 http://logs.openstack.org/65/621465/24/check/grenade-py3/fdef3a8/ : SUCCESS in 57m 38s\n- tempest-full-py3 http://logs.openstack.org/65/621465/24/check/tempest-full-py3/8580339/ : SUCCESS in 1h 23m 25s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/24/check/build-openstack-releasenotes/3cef7e8/html/ : SUCCESS in 13m 58s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/24/check/cinder-tempest-dsvm-lvm-lio-barbican/e32d76c/ : SUCCESS in 1h 55m 44s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/24/check/cinder-tox-bandit-baseline/e8bcd13/ : SUCCESS in 6m 44s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/24/check/nova-multiattach/bc8492f/ : SUCCESS in 59m 46s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/24/check/openstack-tox-functional-py35/8834f29/ : SUCCESS in 4m 40s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/24/check/openstack-tox-functional-py36/414b1ba/ : SUCCESS in 4m 21s\n- cinder-rally-task http://logs.openstack.org/65/621465/24/check/cinder-rally-task/1e9bfc6/ : SUCCESS in 39m 34s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/24/check/openstack-tox-pylint/20d3b01/ : FAILURE in 4m 19s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/24/check/devstack-plugin-ceph-tempest/1186db9/ : SUCCESS in 1h 30m 48s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/24/check/openstack-tox-functional/20a6110/ : SUCCESS in 4m 35s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/24/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/960a705/ : SUCCESS in 1h 02m 52s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/24/check/legacy-tempest-dsvm-lvm-multibackend/5f14625/ : SUCCESS in 1h 17m 01s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/24/check/legacy-tempest-dsvm-full-drbd-devstack/78100c5/ : SUCCESS in 1h 17m 22s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/24/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/62d026f/ : SUCCESS in 1h 16m 41s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/24/check/tempest-slow/7f780a6/ : SUCCESS in 1h 35m 46s","accounts_in_message":[],"_revision_number":24},{"id":"7b417a8347887492cd4ee47dd50c393480ac6645","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-01-04 07:15:47.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/24/check/EMC_Unity_FC/fcf4c7e/EMC_Unity_FC/677 : SUCCESS in 1h 11m 37s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/24/check/EMC_Unity_iSCSI/5be2d45/EMC_Unity_iSCSI/647 : FAILURE in 1h 27m 07s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"23421b80a091ccb95608405f501afca34d462a09","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 07:18:00.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/24/check/NEC-ISCSI/5b86913 : SUCCESS in 2h 03m 29s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/24/check/NEC-FC/d994364 : SUCCESS in 2h 01m 25s","accounts_in_message":[],"_revision_number":24},{"id":"c74a0bdfae0060dd5a6dc12f112628541c06395b","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-04 07:45:54.000000000","message":"Patch Set 24:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/24/check/Cinder-AS13000/215c8e1 : SUCCESS in 2h 29m 30s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/24/check/cinder-test-iscsi/afe09e8 : FAILURE in 35m 20s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/24/check/cinder-test-fc/bf799a2 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":24},{"id":"0a725a576304586d2291989b3bcac4f507f2ff78","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-04 10:43:08.000000000","message":"Patch Set 24:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/24/check-not-vote/ext-cinder-zuul/3f6f8f1 : SUCCESS in 46m 43s","accounts_in_message":[],"_revision_number":24},{"id":"c3bb8fe5f17bf4981735353928797d4ef4701ad0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 12:53:06.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/24/check/EMC_XtremIO_FC/bf15c8d/EMC_XtremIO_FC/1230 : FAILURE in 1h 52m 21s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/24/check/EMC_XtremIO_iSCSI/5c0f282/EMC_XtremIO_iSCSI/1258 : SUCCESS in 1h 20m 40s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":24},{"id":"c33e1c5d3fe50903da2055e133526729c64262c9","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-04 14:26:07.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"d08a99bc43d56a652cf654691a5ec481c0f77316","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 14:28:53.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/25/check/dsvm-tempest-kvm/d86e23c : FAILURE in 23s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/25/check/dsvm-tempest-exe/e57f2ac : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":25},{"id":"8dcc0e96ac136b11cb90d74bb5e851caccc6209a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-04 15:06:16.000000000","message":"Patch Set 25:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/25/check-not-vote/ext-cinder-zuul/5769a30 : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":25},{"id":"7e0e959593fdec955b601598dc0d88e985d08253","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-04 15:19:07.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/25/iscsi/ : SUCCESS in 52m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/25/smb/ : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":25},{"id":"fc0d132f86094411bb13f3a53513b6c14269b6a3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-04 15:29:11.000000000","message":"Patch Set 25:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/25/check-cinder/Cinder-ISER-ISCSI/5d3b8ce : SUCCESS in 1h 01m 41s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/25/check-cinder/Cinder-ISER-LIO/552415b : FAILURE in 56m 19s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":25},{"id":"3d0dada3712816f900432b2ac4fa154e91fab20c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-04 15:30:54.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/25/check/dsvm-tempest-infinibox-fc/6791a93 : SUCCESS in 56m 11s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/25/check/dsvm-tempest-infinibox-iscsi/1d53dbc : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":25},{"id":"852d0cec328b6fd9eb335f2caca658ac05920928","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-04 15:34:13.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/25/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/d7aaa65 : FAILURE in 1h 02m 47s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/25/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/b243671 : SUCCESS in 1h 07m 33s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":25},{"id":"e9f285d0a751ceb88158c6d79cec271d3cc1d3bc","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-04 15:36:57.000000000","message":"Patch Set 25:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/609 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"8826e948211950c63a8fa7b785d75280d92621c3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-04 15:37:43.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/25/check/dsvm-tempest-cisco-zonemanager-job/713ffb0/ : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":25},{"id":"74ca40672310068b076e9d46beb103860cd31e2c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-04 15:39:03.000000000","message":"Patch Set 25:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/25/check/tempest-dsvm-full-xenial/0edbc6f/ : SUCCESS in 1h 12m 34s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/25/check/devstack-plugin-ceph-dsvm-xenial/8f53cbe/ : SUCCESS in 1h 08m 34s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"a586e014ad7976844b561e9195aee08dd4069097","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-04 15:43:47.000000000","message":"Patch Set 25:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/25 : SUCCESS","accounts_in_message":[],"_revision_number":25},{"id":"81f62c4bae519be9f8ad4b140e50098dcb2c6e09","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-04 15:49:35.000000000","message":"Patch Set 25:\n\nBuild Failed \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/610 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"507811554a2f6360273b67fbb2576fbb03a075ac","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-04 16:00:36.000000000","message":"Patch Set 25:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/25/check/dsvm-veritas-access-cinder-driver/bb7fd64 : FAILURE in 1h 22m 54s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/25/check/dsvm-veritas-access-cinder-iscsi-driver/6cb590b : FAILURE in 1h 33m 35s","accounts_in_message":[],"_revision_number":25},{"id":"72e2600794e75431cf67b046de2a3d4759fa2faa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-04 16:34:44.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/25/check/openstack-tox-lower-constraints/aa1416f/ : SUCCESS in 12m 00s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/25/check/openstack-tox-pep8/c4ad844/ : SUCCESS in 7m 00s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/25/check/openstack-tox-py27/06893c6/ : SUCCESS in 9m 14s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/25/check/openstack-tox-py35/ef4b02a/ : SUCCESS in 8m 14s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/25/check/openstack-tox-py36/67678a3/ : SUCCESS in 9m 05s\n- openstack-tox-docs http://logs.openstack.org/65/621465/25/check/openstack-tox-docs/b1e182e/html/ : SUCCESS in 7m 38s\n- requirements-check http://logs.openstack.org/65/621465/25/check/requirements-check/33dc7d0/ : SUCCESS in 3m 10s\n- tempest-full http://logs.openstack.org/65/621465/25/check/tempest-full/333fecd/ : SUCCESS in 1h 22m 03s\n- neutron-grenade http://logs.openstack.org/65/621465/25/check/neutron-grenade/ef4c413/ : SUCCESS in 50m 25s\n- grenade-py3 http://logs.openstack.org/65/621465/25/check/grenade-py3/b2783e8/ : SUCCESS in 55m 41s\n- tempest-full-py3 http://logs.openstack.org/65/621465/25/check/tempest-full-py3/145e40c/ : SUCCESS in 1h 19m 50s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/25/check/build-openstack-releasenotes/8486dc6/html/ : SUCCESS in 10m 12s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/25/check/cinder-tempest-dsvm-lvm-lio-barbican/dc2b3b4/ : SUCCESS in 1h 51m 48s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/25/check/cinder-tox-bandit-baseline/21d0423/ : SUCCESS in 6m 10s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/25/check/nova-multiattach/78db7e6/ : SUCCESS in 59m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/25/check/openstack-tox-functional-py35/3a45f4a/ : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/25/check/openstack-tox-functional-py36/c0d33f6/ : SUCCESS in 4m 54s\n- cinder-rally-task http://logs.openstack.org/65/621465/25/check/cinder-rally-task/49cdae7/ : SUCCESS in 31m 55s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/25/check/openstack-tox-pylint/388c43f/ : FAILURE in 4m 32s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/25/check/devstack-plugin-ceph-tempest/aced8d6/ : FAILURE in 1h 33m 30s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/25/check/openstack-tox-functional/822ac5d/ : SUCCESS in 4m 52s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/25/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/4fdc77b/ : SUCCESS in 57m 38s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/25/check/legacy-tempest-dsvm-lvm-multibackend/0e351a8/ : SUCCESS in 1h 10m 21s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/25/check/legacy-tempest-dsvm-full-drbd-devstack/3b34837/ : SUCCESS in 1h 21m 48s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/25/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/b3c03d9/ : SUCCESS in 1h 25m 02s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/25/check/tempest-slow/0f4ee42/ : SUCCESS in 1h 47m 14s","accounts_in_message":[],"_revision_number":25},{"id":"f14a992e1b1428ef66ba418ffa61d151d90a0cb4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-04 16:44:15.000000000","message":"Patch Set 25:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/25/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/5d3661a : FAILURE in 5m 25s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/25/check/3par-fc-driver-master-client-pip-ssa02-dsvm/014744d : FAILURE in 2m 29s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/25/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/c8ab478 : FAILURE in 32s","accounts_in_message":[],"_revision_number":25},{"id":"07fb6143ce3897c874a63a2fc1b9a69ca2293546","author":{"_account_id":22450,"name":"qnap-ci","email":"openstack-ci@qnap.com","username":"qnap-ci"},"date":"2019-01-04 17:06:08.000000000","message":"Patch Set 25:\n\nBuild failed.  To recheck use \u0027run-QNAP ES CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- qnap-cinder-driver http://openstack-ci-logs.qnap.com/65/621465/25/check/qnap-cinder-driver/0a8180b : FAILURE in 2h 05m 02s","accounts_in_message":[],"_revision_number":25},{"id":"0d87f1e96dc8bb18b6e357afb0bd9362ca920c7a","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-04 17:06:28.000000000","message":"Patch Set 25:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/25/check/Cinder-AS13000/58cb0b2 : SUCCESS in 2h 36m 16s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/25/check/cinder-test-iscsi/f1e0137 : FAILURE in 36m 50s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/25/check/cinder-test-fc/6d8efa8 : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":25},{"id":"31e78af42809e5d19c150197da5aaa4f4379927f","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-01-04 17:34:10.000000000","message":"Patch Set 25:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/25/check/synology-dsm-cinder-iscsi/1e92c81 : FAILURE in 58m 45s","accounts_in_message":[],"_revision_number":25},{"id":"ac0bb024367074f9cbcada51b0b832135a6c8dcb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-04 17:36:44.000000000","message":"Patch Set 25:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-25 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":25},{"id":"419274f34297e3484ef626bcb947f58733268150","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 17:39:17.000000000","message":"Patch Set 25:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/25/check/NEC-ISCSI/0dd6dcd : FAILURE in 3h 01m 24s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/25/check/NEC-FC/24aed9c : SUCCESS in 3h 11m 38s","accounts_in_message":[],"_revision_number":25},{"id":"f2e49dc08c5d738ccb0f72ec14882ccfb62ea8fb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-04 17:43:14.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/25/upstream-check/cinder-cDOT-iSCSI/b3e16ac/ : FAILURE in 29m 35s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/25/upstream-check/cinder-cDOT-NFS/1657a33/ : FAILURE in 22m 51s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/25/upstream-check/cinder-cDOT-FCP/e59ebe5/ : FAILURE in 1h 07m 43s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":25},{"id":"369dc0382eb53ce4a4d26e3ce5f2f7e3f88a6f41","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 17:50:10.000000000","message":"Patch Set 25:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-25-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":25},{"id":"b33ff6fca2936131416d69dcd7104332e5ca764f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-01-04 18:13:31.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/25/check/Dell_SC_FC/3ebde0c/Dell_SC_FC/600 : FAILURE in 1h 01m 28s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/25/check/Dell_SC_iSCSI/d2a6ce7/Dell_SC_iSCSI/589 : SUCCESS in 1h 26m 28s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":25},{"id":"0bb3eabeabe2ec85c021d8fd33e8cbe922b6351f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-01-04 18:14:29.000000000","message":"Patch Set 25:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/25/Lenovo-FC : SUCCESS in 31:51.26\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/25/Lenovo-iSCSI : SUCCESS in 31:48.03","accounts_in_message":[],"_revision_number":25},{"id":"44a821081651366a9101bfa2fe080da6c2652aa0","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-01-04 18:17:14.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/25/check/kaminario-dsvm-tempest-full-iscsi/1c71c25 : SUCCESS in 1h 41m 52s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/25/check/kaminario-dsvm-tempest-full-FC/01d8f09 : SUCCESS in 1h 36m 18s","accounts_in_message":[],"_revision_number":25},{"id":"4268f3cd33b5244476f9eb522aa9305f2879a2ce","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-01-04 18:21:39.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/25/upstream-solidfire-check/cinder-solidfire-iSCSI/f41655d/ : FAILURE in 1h 44m 33s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":25},{"id":"fe648ba40a0b5f6963551e750af70cddb657674e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-01-04 18:27:38.000000000","message":"Patch Set 25:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/25/HPMSA-FC : SUCCESS in 36:00.05\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/25/HPMSA-iSCSI : SUCCESS in 36:15.16","accounts_in_message":[],"_revision_number":25},{"id":"4580e69289a7f7f4036ac48f9f8974be861f1d57","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 18:27:59.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/65/621465/25/check/EMC_VxFlexOS/6b3a753/EMC_VxFlexOS/743 : SUCCESS in 1h 52m 53s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":25},{"id":"997aadd6eb3560c5698c0a9cab12fae77bddddbb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-01-04 18:35:50.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/25/check/EMC_Unity_FC/cc75914/EMC_Unity_FC/681 : FAILURE in 2h 55m 20s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/25/check/EMC_Unity_iSCSI/98a39b0/EMC_Unity_iSCSI/651 : FAILURE in 1h 31m 12s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"000d9e261ca02544da2d7b047c0f1faaebfabf59","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 19:33:52.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/65/621465/25/upstream-eseries-check/cinder-eseries-iSCSI/b46a137/ : FAILURE in 2h 57m 34s (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/65/621465/25/upstream-eseries-check/cinder-eseries-FCP/967bdfb/ : FAILURE in 45m 51s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":25},{"id":"3750f3a282eec16c97658aca1031c6569f0c9dbb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 19:40:19.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VNX_iSCSI http://publiclogs.emc.com/621465/25/EMC_VNX_iSCSI/352 : FAILURE in 2h 51m 07s\n- EMC_VNX_FC http://publiclogs.emc.com/621465/25/EMC_VNX_FC/322 : ABORTED in 3h 05m 18s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":25},{"id":"6bb99bc548ca12e50a62bac78c258fc1a14a0650","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-01-04 20:26:01.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/25/check/EMC_XtremIO_FC/6a66394/EMC_XtremIO_FC/1234 : FAILURE in 1h 52m 53s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/25/check/EMC_XtremIO_iSCSI/cb98fee/EMC_XtremIO_iSCSI/1262 : SUCCESS in 1h 20m 38s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":25},{"id":"b4f38fee1353ef210b2f7d0136bcb5c81dd48f89","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-01-04 22:13:46.000000000","message":"Patch Set 25: Code-Review-1 Review-Priority+2\n\n(13 comments)\n\nThis looks pretty close.  There are a number of convention/formatting issues to address but think they can be fixed.  Also we need to see the CI start reporting successfully.\n\nThanks!\nJay","accounts_in_message":[],"_revision_number":25},{"id":"73082b7339d909df2eb1cce4c5645368af53ca79","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-07 03:45:04.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"732bf51e82ae694d1920af7a9116cf4827224ec8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-07 03:47:42.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/26/check/dsvm-tempest-kvm/ff3b55b : FAILURE in 22s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/26/check/dsvm-tempest-exe/89c99d9 : FAILURE in 2s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":26},{"id":"ec88ed7d009b9e8b39aea2a897c34bf6540cacae","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-07 03:48:23.000000000","message":"Patch Set 25:\n\n(13 comments)\n\nThanks Jay, reviews are all addressed.\n\nCI is still on the way..","accounts_in_message":[],"_revision_number":25},{"id":"6ca4404d301605104ffe09ce5101823d10bcbb9d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-01-07 04:21:35.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/26/check/EMC_XtremIO_FC/558529c/EMC_XtremIO_FC/1251 : FAILURE in 36m 08s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/26/check/EMC_XtremIO_iSCSI/b186889/EMC_XtremIO_iSCSI/1282 : FAILURE in 28m 20s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":26},{"id":"39e255beeacb496f44d1d0c3d2ed59d23dcca15c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-07 04:37:36.000000000","message":"Patch Set 26:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/26/check-not-vote/ext-cinder-zuul/bdcaca5 : SUCCESS in 51m 37s","accounts_in_message":[],"_revision_number":26},{"id":"5ef1a783bb9ea8affb7f407052393d572565d24d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-07 04:45:57.000000000","message":"Patch Set 26:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/26/check-cinder/Cinder-ISER-ISCSI/d5f0325 : SUCCESS in 59m 37s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/26/check-cinder/Cinder-ISER-LIO/84c9acc : FAILURE in 57m 01s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":26},{"id":"043b8c5bf022be3ba97c7c4fe0f51b16044fa5c7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-07 04:54:22.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/26/check/dsvm-tempest-infinibox-fc/2f7990c : SUCCESS in 1h 08m 25s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/26/check/dsvm-tempest-infinibox-iscsi/51ff597 : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":26},{"id":"91c9af56936f7b47abce8e9ee1d84c8fae050ebf","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-07 05:01:39.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/26 : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"fe7acbd26e77f059101155b1c1ce4651cba0a96c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-07 05:07:46.000000000","message":"Patch Set 26:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/26/iscsi/ : FAILURE in 1h 22m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/26/smb/ : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":26},{"id":"e43b1aef60d7a00c6e1143d87fd0374a8a59c055","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-07 05:12:25.000000000","message":"Patch Set 26:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/26/check/tempest-dsvm-full-xenial/159e747/ : SUCCESS in 1h 03m 02s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/26/check/devstack-plugin-ceph-dsvm-xenial/032416c/ : SUCCESS in 1h 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"95e7f3ced3f6acd90a26d1d933acbf2b875fc5a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-07 05:40:33.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/26/check/openstack-tox-lower-constraints/ccadd9d/ : SUCCESS in 12m 13s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/26/check/openstack-tox-pep8/c23d208/ : SUCCESS in 8m 12s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/26/check/openstack-tox-py27/c014886/ : SUCCESS in 10m 32s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/26/check/openstack-tox-py35/f74b3f0/ : SUCCESS in 10m 23s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/26/check/openstack-tox-py36/578232f/ : SUCCESS in 9m 14s\n- openstack-tox-docs http://logs.openstack.org/65/621465/26/check/openstack-tox-docs/82d8417/html/ : SUCCESS in 6m 44s\n- requirements-check http://logs.openstack.org/65/621465/26/check/requirements-check/6e8627d/ : SUCCESS in 2m 43s\n- tempest-full http://logs.openstack.org/65/621465/26/check/tempest-full/ce6bddd/ : SUCCESS in 1h 20m 44s\n- neutron-grenade http://logs.openstack.org/65/621465/26/check/neutron-grenade/379bcb5/ : SUCCESS in 49m 34s\n- grenade-py3 http://logs.openstack.org/65/621465/26/check/grenade-py3/744eba6/ : FAILURE in 55m 31s\n- tempest-full-py3 http://logs.openstack.org/65/621465/26/check/tempest-full-py3/887752e/ : SUCCESS in 1h 18m 18s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/26/check/build-openstack-releasenotes/48e43ce/html/ : SUCCESS in 10m 58s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/26/check/cinder-tempest-dsvm-lvm-lio-barbican/7a71eb3/ : FAILURE in 1h 51m 11s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/26/check/cinder-tox-bandit-baseline/864d18a/ : SUCCESS in 6m 07s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/26/check/nova-multiattach/24b9903/ : FAILURE in 59m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/26/check/openstack-tox-functional-py35/58ef699/ : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/26/check/openstack-tox-functional-py36/7449c6e/ : SUCCESS in 4m 34s\n- cinder-rally-task http://logs.openstack.org/65/621465/26/check/cinder-rally-task/95ab76b/ : SUCCESS in 38m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/26/check/openstack-tox-pylint/0b01883/ : FAILURE in 4m 27s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/26/check/devstack-plugin-ceph-tempest/198c70f/ : SUCCESS in 1h 43m 10s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/26/check/openstack-tox-functional/819b57d/ : SUCCESS in 4m 09s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/26/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/2521a1c/ : SUCCESS in 1h 03m 20s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/26/check/legacy-tempest-dsvm-lvm-multibackend/c36a07c/ : SUCCESS in 1h 20m 37s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/26/check/legacy-tempest-dsvm-full-drbd-devstack/3fa1494/ : SUCCESS in 1h 22m 35s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/26/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/23d92c0/ : SUCCESS in 1h 28m 02s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/26/check/tempest-slow/38323e8/ : SUCCESS in 1h 45m 10s","accounts_in_message":[],"_revision_number":26},{"id":"a37e7dfbabac2dd447d299817d5c388cb773e82f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-07 05:48:36.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/26/check/dsvm-tempest-cisco-zonemanager-job/7ab70db/ : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":26},{"id":"c870737fe57b0584cdb63a4550a00f8552128a54","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-07 05:57:43.000000000","message":"Patch Set 26:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/26/check/dsvm-veritas-access-cinder-driver/1404f7f : FAILURE in 1h 19m 28s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/26/check/dsvm-veritas-access-cinder-iscsi-driver/cb37905 : FAILURE in 2h 11m 54s","accounts_in_message":[],"_revision_number":26},{"id":"82a8685cfc8b8b3c73b35c72ea3e870406ae265a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-07 06:01:22.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/26/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/fac5f27 : SUCCESS in 44m 09s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/26/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/044c283 : FAILURE in 1h 49m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":26},{"id":"1f4d2ba585c616de69348861884e859c74d2d16e","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-07 07:44:23.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"eac1069b41a380399a863db6e3611645836a18ad","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-07 07:46:22.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/26/check/dsvm-tempest-kvm/b3d2586 : FAILURE in 2s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/26/check/dsvm-tempest-exe/3f5c3c2 : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":26},{"id":"4cf1e5ec70d5004d59ae0aee943d1440ba57872a","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-07 08:09:10.000000000","message":"Patch Set 26:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/26/check/Cinder-AS13000/978504e : SUCCESS in 2h 29m 11s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/26/check/cinder-test-iscsi/dd434e3 : FAILURE in 31m 27s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/26/check/cinder-test-fc/da59eb2 : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":26},{"id":"3088368ffdcdad86124ea0e801d25dfff6e62330","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-07 08:31:44.000000000","message":"Patch Set 26:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/26/check-not-vote/ext-cinder-zuul/fdb91b8 : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":26},{"id":"57849fb91db05c01bfce8973e98c195a29a3a8d4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-07 08:34:56.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/26/iscsi/ : SUCCESS in 50m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/26/smb/ : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":26},{"id":"f970e762bffb6c4254338dbc88aa15a2f16ce65d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-07 08:55:22.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/26/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/169c6aa : SUCCESS in 46m 02s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/26/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/b92aca2 : SUCCESS in 1h 09m 53s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":26},{"id":"b598ebb776e6df3f9bc3c5bc75497e328c77a3c9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-07 08:55:34.000000000","message":"Patch Set 26:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/26/check/tempest-dsvm-full-xenial/3934f29/ : SUCCESS in 1h 10m 54s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/26/check/devstack-plugin-ceph-dsvm-xenial/78c2a0b/ : SUCCESS in 1h 06m 49s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"942b6df36f1d8f910a5787e9cba4a4a472d04627","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-07 08:55:58.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/26/check/dsvm-tempest-cisco-zonemanager-job/42e5ef5/ : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":26},{"id":"98cafdf5d7d35e0454fe2d85035b550134ade195","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-07 09:03:09.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/26/check/NEC-ISCSI/e142e67 : SUCCESS in 1h 47m 06s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/26/check/NEC-FC/247d1a5 : SUCCESS in 1h 52m 50s","accounts_in_message":[],"_revision_number":26},{"id":"0e6601874205767542cdc96ec82e4074c1d1052f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-07 10:52:13.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/26/check/openstack-tox-lower-constraints/80baa35/ : SUCCESS in 9m 25s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/26/check/openstack-tox-pep8/3a3269a/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/26/check/openstack-tox-py27/3ef474a/ : SUCCESS in 8m 39s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/26/check/openstack-tox-py35/7859279/ : SUCCESS in 9m 22s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/26/check/openstack-tox-py36/7df2657/ : SUCCESS in 8m 11s\n- openstack-tox-docs http://logs.openstack.org/65/621465/26/check/openstack-tox-docs/116e014/html/ : SUCCESS in 7m 25s\n- requirements-check http://logs.openstack.org/65/621465/26/check/requirements-check/ba3c26c/ : SUCCESS in 2m 58s\n- tempest-full http://logs.openstack.org/65/621465/26/check/tempest-full/5fcd73e/ : SUCCESS in 1h 22m 03s\n- neutron-grenade http://logs.openstack.org/65/621465/26/check/neutron-grenade/d8467a2/ : SUCCESS in 1h 09m 13s\n- grenade-py3 http://logs.openstack.org/65/621465/26/check/grenade-py3/5cc2d38/ : SUCCESS in 1h 04m 46s\n- tempest-full-py3 http://logs.openstack.org/65/621465/26/check/tempest-full-py3/c847314/ : SUCCESS in 1h 18m 39s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/26/check/build-openstack-releasenotes/c18f670/html/ : SUCCESS in 14m 25s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/26/check/cinder-tempest-dsvm-lvm-lio-barbican/7eedea2/ : SUCCESS in 2h 10m 57s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/26/check/cinder-tox-bandit-baseline/7975846/ : SUCCESS in 6m 42s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/26/check/nova-multiattach/89d5a45/ : SUCCESS in 1h 02m 36s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/26/check/openstack-tox-functional-py35/ba19340/ : SUCCESS in 4m 33s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/26/check/openstack-tox-functional-py36/f8b0148/ : SUCCESS in 4m 45s\n- cinder-rally-task http://logs.openstack.org/65/621465/26/check/cinder-rally-task/4302ffb/ : SUCCESS in 37m 59s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/26/check/openstack-tox-pylint/30558b6/ : FAILURE in 4m 22s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/26/check/devstack-plugin-ceph-tempest/6b9b223/ : SUCCESS in 1h 35m 09s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/26/check/openstack-tox-functional/a4becdd/ : SUCCESS in 5m 01s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/26/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/b3755ce/ : SUCCESS in 1h 06m 23s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/26/check/legacy-tempest-dsvm-lvm-multibackend/e593f5d/ : SUCCESS in 1h 18m 44s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/26/check/legacy-tempest-dsvm-full-drbd-devstack/7388b68/ : SUCCESS in 1h 39m 51s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/26/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/21ccddb/ : TIMED_OUT in 3h 01m 03s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/26/check/tempest-slow/e1ca99f/ : SUCCESS in 1h 41m 00s","accounts_in_message":[],"_revision_number":26},{"id":"6c6dcfcb131915f9ae30a5dfd2eb8235b8993930","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-07 10:54:53.000000000","message":"Patch Set 26:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/26/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/704a4aa : NOT_REGISTERED\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/26/check/3par-fc-driver-master-client-pip-ssa02-dsvm/b9e58ba : NOT_REGISTERED\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/26/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/8fde761 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":26},{"id":"acf1b73d23c99fc0818bd9fcb78491b58d0ceb24","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-01-07 11:30:09.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/26/upstream-solidfire-check/cinder-solidfire-iSCSI/9d803ea/ : FAILURE in 36m 50s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":26},{"id":"b2e586336c03f0f1bfc7f3fdc29a096b5b365e6d","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-01-07 11:43:53.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/26/check/synology-dsm-cinder-iscsi/14de91e : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":26},{"id":"0ef8895c065cb16fbeb8029be38ef34e08dc744c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-07 11:52:08.000000000","message":"Patch Set 26:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-26 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":26},{"id":"f805b93abaf4b04d7a335059ed9ba792f2619859","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-07 11:59:46.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/26/upstream-check/cinder-cDOT-iSCSI/2c59104/ : FAILURE in 27m 46s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/26/upstream-check/cinder-cDOT-NFS/ea288bb/ : FAILURE in 1h 06m 53s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/26/upstream-check/cinder-cDOT-FCP/ed249de/ : FAILURE in 49m 35s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":26},{"id":"cc1d6b87e243dcc24c2152bd4681b7cd605d487f","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-01-07 12:06:53.000000000","message":"Patch Set 26:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-26-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"7efbeb4fc6208d2e080b18e78831bf73fd94ed20","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-01-07 12:29:42.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/26/check/Dell_SC_FC/b33ec8a/Dell_SC_FC/606 : FAILURE in 57m 31s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/26/check/Dell_SC_iSCSI/2c182c1/Dell_SC_iSCSI/595 : SUCCESS in 1h 37m 10s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":26},{"id":"001c94ae33c111898cda0f7dba9388eee3c201cc","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-01-07 12:32:06.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/26/check/kaminario-dsvm-tempest-full-iscsi/04b946c : SUCCESS in 1h 39m 21s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/26/check/kaminario-dsvm-tempest-full-FC/a2158e0 : SUCCESS in 1h 38m 09s","accounts_in_message":[],"_revision_number":26},{"id":"583842b78d48f972c952355710deebdf5ba4af56","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-01-07 12:33:30.000000000","message":"Patch Set 26:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/26/HPMSA-FC : SUCCESS in 36:14.27\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/26/HPMSA-iSCSI : SUCCESS in 36:15.29","accounts_in_message":[],"_revision_number":26},{"id":"b7e491a2063ca2dee48e31bad878da185ff23b0d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-01-07 12:34:55.000000000","message":"Patch Set 26:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/26/Lenovo-FC : SUCCESS in 36:56.72\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/26/Lenovo-iSCSI : SUCCESS in 36:33.35","accounts_in_message":[],"_revision_number":26},{"id":"954bbe0074c0da088a6c55aafc99b8807440ef54","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-01-07 12:39:33.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/65/621465/26/check/EMC_VxFlexOS/25edfaf/EMC_VxFlexOS/750 : SUCCESS in 1h 47m 01s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":26},{"id":"3c68162ce46bd4f7b60fcf8adba1a50be0bd281e","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-01-07 12:40:15.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/65/621465/26/upstream-eseries-check/cinder-eseries-iSCSI/bc618d3/ : FAILURE in 1h 47m 07s (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/65/621465/26/upstream-eseries-check/cinder-eseries-FCP/d4c8ab6/ : FAILURE in 47m 53s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":26},{"id":"63f1d38d408e56b4b10867b0b0de356dbe2ffd1d","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-08 06:29:16.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"3d648d92c341c57e12ec273b72de7a1464b41d57","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-08 06:30:53.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/27/check/dsvm-tempest-kvm/855a63b : FAILURE in 4s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/27/check/dsvm-tempest-exe/8a89c1f : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":27},{"id":"1a5c7abc95c84de5d18d7f9223983a53a9fa0d77","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-08 07:19:31.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/27/iscsi/ : SUCCESS in 49m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/27/smb/ : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":27},{"id":"0a74d2286fe5936285737d543c6af13bfb6d3a2d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-08 07:19:56.000000000","message":"Patch Set 27:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/27/check/fujitsu-eternus-dx-iscsi/0b52c5b : FAILURE in 44m 10s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/27/check/fujitsu-eternus-dx-fc/0a65bbb : FAILURE in 42m 50s","accounts_in_message":[],"_revision_number":27},{"id":"9b22a2a6b9d9e13c31ee6780a5df462be6ecd46f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-08 07:33:06.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/27/check-cinder/Cinder-ISER-ISCSI/92796a1 : SUCCESS in 1h 02m 27s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/27/check-cinder/Cinder-ISER-LIO/680faba : SUCCESS in 55m 21s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":27},{"id":"a5de3bad32af0bb500fd589e8a57be33b7013032","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-08 07:40:49.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/27/check/dsvm-tempest-cisco-zonemanager-job/54c3c4a/ : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":27},{"id":"2aa356e2e7271ac7f9b0f9c0390590b126f301ac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-08 07:50:58.000000000","message":"Patch Set 27:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/27/check/tempest-dsvm-full-xenial/8b6961c/ : SUCCESS in 1h 21m 18s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/27/check/devstack-plugin-ceph-dsvm-xenial/b703548/ : SUCCESS in 1h 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"bd66d17be9ae57ad36bdce6ea47ace631d53fc5c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-08 07:53:06.000000000","message":"Patch Set 27:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/27 : SUCCESS","accounts_in_message":[],"_revision_number":27},{"id":"21038d163d8293027172af949618bd41bcee9331","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-08 08:18:40.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/27/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/b6b208d : SUCCESS in 45m 55s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/27/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/5763f51 : FAILURE in 1h 49m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":27},{"id":"90cf196b2786b438ea3906641c7d6811420a00af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 08:52:05.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/27/check/openstack-tox-lower-constraints/c45d5a2/ : SUCCESS in 12m 29s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/27/check/openstack-tox-pep8/ddf56c3/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/27/check/openstack-tox-py27/7c3a714/ : SUCCESS in 7m 53s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/27/check/openstack-tox-py35/ae1ed2d/ : SUCCESS in 9m 24s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/27/check/openstack-tox-py36/8b39191/ : SUCCESS in 9m 13s\n- openstack-tox-docs http://logs.openstack.org/65/621465/27/check/openstack-tox-docs/d491d98/html/ : SUCCESS in 8m 08s\n- requirements-check http://logs.openstack.org/65/621465/27/check/requirements-check/5c20be3/ : SUCCESS in 2m 51s\n- tempest-full http://logs.openstack.org/65/621465/27/check/tempest-full/5fcb103/ : SUCCESS in 1h 25m 33s\n- neutron-grenade http://logs.openstack.org/65/621465/27/check/neutron-grenade/8da0829/ : SUCCESS in 56m 07s\n- grenade-py3 http://logs.openstack.org/65/621465/27/check/grenade-py3/bcb4f67/ : SUCCESS in 59m 31s\n- tempest-full-py3 http://logs.openstack.org/65/621465/27/check/tempest-full-py3/fad107f/ : SUCCESS in 1h 32m 10s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/27/check/build-openstack-releasenotes/95bdd97/html/ : SUCCESS in 13m 08s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/27/check/cinder-tempest-dsvm-lvm-lio-barbican/0b88c0a/ : SUCCESS in 2h 14m 39s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/27/check/cinder-tox-bandit-baseline/fa6dd3f/ : SUCCESS in 7m 51s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/27/check/nova-multiattach/c7d683c/ : SUCCESS in 55m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/27/check/openstack-tox-functional-py35/249fce6/ : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/27/check/openstack-tox-functional-py36/4c1af7b/ : SUCCESS in 5m 13s\n- cinder-rally-task http://logs.openstack.org/65/621465/27/check/cinder-rally-task/c5972fe/ : SUCCESS in 34m 26s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/27/check/openstack-tox-pylint/7ff5559/ : FAILURE in 5m 28s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/27/check/devstack-plugin-ceph-tempest/8a5843c/ : SUCCESS in 1h 40m 32s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/27/check/openstack-tox-functional/76405a0/ : SUCCESS in 5m 19s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/27/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/839842f/ : SUCCESS in 1h 02m 20s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/27/check/legacy-tempest-dsvm-lvm-multibackend/f9b449d/ : SUCCESS in 1h 12m 21s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/27/check/legacy-tempest-dsvm-full-drbd-devstack/e8668b4/ : SUCCESS in 1h 23m 23s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/27/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/fcbbc99/ : SUCCESS in 1h 15m 48s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/27/check/tempest-slow/4ab0d32/ : SUCCESS in 1h 57m 05s","accounts_in_message":[],"_revision_number":27},{"id":"98193ff4c3b13178147c64d1159b4fc6db62cd04","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-08 08:55:47.000000000","message":"Patch Set 27:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/27/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8eb4de0 : NOT_REGISTERED\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/27/check/3par-fc-driver-master-client-pip-ssa02-dsvm/9fb2378 : NOT_REGISTERED\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/27/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/d77b691 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":27},{"id":"1852bd8ea8c66a1e7c905abd5872b3edd5bb4ae2","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-08 09:06:50.000000000","message":"Patch Set 27:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/27/check/Cinder-AS13000/e6643f8 : UNSTABLE in 2h 34m 51s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/27/check/cinder-test-iscsi/591eb5c : ABORTED in 28m 09s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/27/check/cinder-test-fc/c4ee32f : FAILURE in 26m 11s","accounts_in_message":[],"_revision_number":27},{"id":"ad8b3e9f0e4cb9d16de94d465baf6d21d0f5078f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-01-08 09:30:05.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/27/upstream-solidfire-check/cinder-solidfire-iSCSI/70eb5dd/ : FAILURE in 36m 49s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":27},{"id":"f1ac06ac13c2931230367df0ee925ee749c8acf9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-08 09:42:13.000000000","message":"Patch Set 27:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-27 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":27},{"id":"ed005e24b29f65dc2e88e93601749ac1801aec7c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-08 09:46:34.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/27/upstream-check/cinder-cDOT-iSCSI/e70e4db/ : FAILURE in 52m 41s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/27/upstream-check/cinder-cDOT-NFS/7c6a449/ : FAILURE in 26m 54s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/27/upstream-check/cinder-cDOT-FCP/0faba05/ : FAILURE in 48m 30s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":27},{"id":"6098315be2e71394b1264197e34c9502ec42bd20","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-01-08 09:46:43.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/27/check/synology-dsm-cinder-iscsi/56f17da : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":27},{"id":"e86fac12efcfc367c82bdf623062e6264e65534d","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-08 10:09:42.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/27/check/NEC-ISCSI/e715384 : SUCCESS in 1h 37m 25s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/27/check/NEC-FC/3c2a60b : SUCCESS in 1h 53m 11s","accounts_in_message":[],"_revision_number":27},{"id":"babb20af56ae01c0eb5e8dc7b692607a4fd5e730","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-01-08 10:16:38.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/27/check/Dell_SC_FC/46304e1/Dell_SC_FC/613 : SUCCESS in 1h 10m 26s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/27/check/Dell_SC_iSCSI/ac959d7/Dell_SC_iSCSI/602 : SUCCESS in 1h 21m 18s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":27},{"id":"5c3e90cc7d44651a6ab6a7e60ad88095dc88f305","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-01-08 10:29:09.000000000","message":"Patch Set 27:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/27/HPMSA-FC : SUCCESS in 33:38.38\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/27/HPMSA-iSCSI : SUCCESS in 34:08.59","accounts_in_message":[],"_revision_number":27},{"id":"3ff146377433f4efcad476ff5b177cbc8a061d09","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-01-08 10:29:10.000000000","message":"Patch Set 27:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/27/Lenovo-FC : SUCCESS in 34:17.23\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/27/Lenovo-iSCSI : SUCCESS in 33:31.17","accounts_in_message":[],"_revision_number":27},{"id":"3a3ebda9ffa663adc308ef77443a4e0ae32cb4b2","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-01-08 11:17:06.000000000","message":"Patch Set 27:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-27-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":27},{"id":"03ffdc93612775ef551be212a2c998fff6851066","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-01-08 13:10:44.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/65/621465/27/upstream-eseries-check/cinder-eseries-iSCSI/8768b1b/ : FAILURE in 1h 48m 33s (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/65/621465/27/upstream-eseries-check/cinder-eseries-FCP/cd15e53/ : FAILURE in 42m 56s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":27},{"id":"c931d90a3827140766a6bd0116f78a6f1804883c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-08 14:02:08.000000000","message":"Patch Set 27:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/27/check-not-vote/ext-cinder-zuul/d71bacc : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":27},{"id":"a71c865811b097d9e95155f2862c850fd519fdbf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-08 16:27:43.000000000","message":"Patch Set 27:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/27/check/dsvm-tempest-infinibox-fc/c641ad5 : FAILURE in 13m 39s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/27/check/dsvm-tempest-infinibox-iscsi/ec8f70b : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":27},{"id":"cf24b6d424c1403d58070aa4705205cdfb3afc90","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-01-08 16:47:13.000000000","message":"Patch Set 27: Code-Review+1\n\nThank you for addressing my comments.  I think this can be ready to go if we get the CI running.","accounts_in_message":[],"_revision_number":27},{"id":"4587c11d960bcb138c1ac04c31063b76bf3de85f","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-08 19:06:34.000000000","message":"Patch Set 27:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/27/check/dsvm-veritas-access-cinder-driver/04a05b1 : FAILURE in 13s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/27/check/dsvm-veritas-access-cinder-iscsi-driver/7879098 : FAILURE in 11s","accounts_in_message":[],"_revision_number":27},{"id":"a3bc53649667cfc5b8f3ce9dc31ac256cb2e2d24","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-09 01:56:42.000000000","message":"Patch Set 27:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/27 : FAILED","accounts_in_message":[],"_revision_number":27},{"id":"8e7243facb6d0453300e4eab20d2ee2353f09ab1","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-09 08:30:42.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"4afe9b52c96107d3221ddcf53a52a613cebe4e5e","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-09 08:32:43.000000000","message":"Patch Set 28:\n\nAdded retry logic for volume attachment.","accounts_in_message":[],"_revision_number":28},{"id":"aace9ea61be7486d036afd83ce2d7692a3521b9d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-09 08:33:12.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/28/check/dsvm-tempest-kvm/4db2621 : FAILURE in 24s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/28/check/dsvm-tempest-exe/a2f16af : FAILURE in 2s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":28},{"id":"af2d491974c7598b4e1cd5f0ea348233137546e5","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-09 08:54:55.000000000","message":"Patch Set 28:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/28/check/fujitsu-eternus-dx-iscsi/dfb0f00 : FAILURE in 9m 08s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/28/check/fujitsu-eternus-dx-fc/796a483 : FAILURE in 12m 18s","accounts_in_message":[],"_revision_number":28},{"id":"4767e7acc123154bd95441e4eb24cc744dd38487","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-09 09:12:05.000000000","message":"Patch Set 28:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/28/check-not-vote/ext-cinder-zuul/a4a7b3b : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":28},{"id":"6c22f70ea31b7d51d576ef23b548b1f52f2950eb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-09 09:20:43.000000000","message":"Patch Set 28:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/28/iscsi/ : FAILURE in 49m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/28/smb/ : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":28},{"id":"6e0f0760a5b303004ddba040998b925bea65ba74","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-09 09:32:10.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/28/check-cinder/Cinder-ISER-ISCSI/bbb2783 : SUCCESS in 59m 08s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/28/check-cinder/Cinder-ISER-LIO/8d1d2e0 : SUCCESS in 59m 52s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":28},{"id":"d1263ac305e4bfd2247a8d7b56f7055cd2d0064a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-09 09:42:43.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/28/check/dsvm-tempest-cisco-zonemanager-job/62c9bab/ : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":28},{"id":"88466b017f784264d06f405ee99d9a5d4e8fa544","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-09 09:47:20.000000000","message":"Patch Set 28:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/28/check/tempest-dsvm-full-xenial/1a66c09/ : SUCCESS in 1h 12m 08s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/28/check/devstack-plugin-ceph-dsvm-xenial/cb872a9/ : SUCCESS in 1h 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"96ed6b987f2907a9d01ec8640df3b3fb3dcebee3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-09 09:49:16.000000000","message":"Patch Set 28:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/28 : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"d7a4ce9e3e3f97b16d4929bed93acd1687173380","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-09 09:52:13.000000000","message":"Patch Set 28:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/28 : FAILED","accounts_in_message":[],"_revision_number":28},{"id":"caad52d5b63f7e6bb8c6d5f0a3a4a1d431211d00","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-09 09:57:38.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/28/check/dsvm-tempest-infinibox-fc/87fbc8b : SUCCESS in 1h 04m 48s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/28/check/dsvm-tempest-infinibox-iscsi/025fc1b : SUCCESS in 1h 25m 38s","accounts_in_message":[],"_revision_number":28},{"id":"c7e91cf440106ebbdad5bccadbc20a7c37ea8802","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-09 10:10:12.000000000","message":"Patch Set 28:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/28/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/3d4af1c : SUCCESS in 44m 47s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/28/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e76df23 : FAILURE in 1h 39m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":28},{"id":"87b2b5d28ff48cc42f583242ad9a9791aa35bf9e","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-09 10:13:21.000000000","message":"Patch Set 28:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/28/check/dsvm-veritas-access-cinder-driver/6b2dc2d : FAILURE in 25s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/28/check/dsvm-veritas-access-cinder-iscsi-driver/2c31ec6 : FAILURE in 1h 41m 40s","accounts_in_message":[],"_revision_number":28},{"id":"1b478ee9bae2cc80fa7ce0965904a45bdab3adc3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-09 10:36:09.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/28/check/NEC-ISCSI/c77f978 : SUCCESS in 2h 04m 32s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/28/check/NEC-FC/9b8ebfd : SUCCESS in 1h 58m 06s","accounts_in_message":[],"_revision_number":28},{"id":"5ea79f83756daa8229a7cd07a23c689d780a788c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 10:37:35.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/28/check/openstack-tox-lower-constraints/38f11c6/ : SUCCESS in 11m 16s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/28/check/openstack-tox-pep8/b4a0b6d/ : SUCCESS in 6m 58s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/28/check/openstack-tox-py27/20c7d38/ : SUCCESS in 9m 02s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/28/check/openstack-tox-py35/6d6a224/ : SUCCESS in 9m 13s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/28/check/openstack-tox-py36/f8a0b3f/ : SUCCESS in 9m 08s\n- openstack-tox-docs http://logs.openstack.org/65/621465/28/check/openstack-tox-docs/4db0abb/html/ : SUCCESS in 7m 47s\n- requirements-check http://logs.openstack.org/65/621465/28/check/requirements-check/d902a83/ : SUCCESS in 2m 34s\n- tempest-full http://logs.openstack.org/65/621465/28/check/tempest-full/8651b1a/ : SUCCESS in 1h 26m 23s\n- neutron-grenade http://logs.openstack.org/65/621465/28/check/neutron-grenade/a9c2d58/ : SUCCESS in 57m 58s\n- grenade-py3 http://logs.openstack.org/65/621465/28/check/grenade-py3/b7785d3/ : SUCCESS in 1h 01m 10s\n- tempest-full-py3 http://logs.openstack.org/65/621465/28/check/tempest-full-py3/6e1310d/ : SUCCESS in 1h 04m 22s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/28/check/build-openstack-releasenotes/58e9c2e/html/ : SUCCESS in 13m 34s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/28/check/cinder-tempest-dsvm-lvm-lio-barbican/dc415ad/ : SUCCESS in 1h 47m 56s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/28/check/cinder-tox-bandit-baseline/1fc6a0f/ : SUCCESS in 7m 29s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/28/check/nova-multiattach/cf38eba/ : SUCCESS in 1h 05m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/28/check/openstack-tox-functional-py35/465656e/ : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/28/check/openstack-tox-functional-py36/f988edf/ : SUCCESS in 5m 04s\n- cinder-rally-task http://logs.openstack.org/65/621465/28/check/cinder-rally-task/2efcf02/ : SUCCESS in 40m 51s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/28/check/openstack-tox-pylint/3c15bf6/ : FAILURE in 4m 48s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/28/check/devstack-plugin-ceph-tempest/65e067d/ : SUCCESS in 1h 33m 15s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/28/check/openstack-tox-functional/ff939ee/ : SUCCESS in 5m 13s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/28/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/30eec87/ : SUCCESS in 1h 02m 38s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-lvm-multibackend/ecfef9d/ : SUCCESS in 1h 16m 28s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-full-drbd-devstack/db3c27f/ : SUCCESS in 1h 20m 29s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/38d22d9/ : SUCCESS in 1h 26m 10s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/28/check/tempest-slow/7f5c5ee/ : SUCCESS in 1h 59m 31s","accounts_in_message":[],"_revision_number":28},{"id":"d77ae7ca9b5da0c41e257cd76cff1fe4154c4c31","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-09 10:46:13.000000000","message":"Patch Set 28:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/28/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f4280d4 : FAILURE in 5m 25s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/28/check/3par-fc-driver-master-client-pip-ssa02-dsvm/ce96336 : FAILURE in 2m 25s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/28/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/0cc1817 : FAILURE in 26s","accounts_in_message":[],"_revision_number":28},{"id":"0f7529e193fd842d38f984801c48b82136d8d209","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-01-09 10:54:13.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/28/upstream-solidfire-check/cinder-solidfire-iSCSI/6703772/ : FAILURE in 14m 54s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":28},{"id":"50caaa53997b96ca3f5ca1d4fbf5fd5b0a4e70df","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-01-09 11:08:45.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/65/621465/28/upstream-eseries-check/cinder-eseries-iSCSI/6284f6b/ : FAILURE in 7m 29s (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/65/621465/28/upstream-eseries-check/cinder-eseries-FCP/f5565bd/ : FAILURE in 29m 44s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":28},{"id":"a17a234de5d028dae1cecd625a9b73ae22563bff","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-09 11:20:24.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/28/upstream-check/cinder-cDOT-iSCSI/1d01670/ : FAILURE in 28m 13s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/28/upstream-check/cinder-cDOT-NFS/f4c1320/ : FAILURE in 28m 05s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/28/upstream-check/cinder-cDOT-FCP/21014fe/ : FAILURE in 41m 54s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":28},{"id":"31c29bac6c510c439715214d8cf21493a32df707","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-09 11:28:51.000000000","message":"Patch Set 28:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-28 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":28},{"id":"d055454c920adc6b2002898ed38324d156cd1daa","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-01-09 11:31:39.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/28/check/synology-dsm-cinder-iscsi/03057f9 : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":28},{"id":"8a29e134242e249dfbea0b2a31108912048fb415","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-01-09 11:51:59.000000000","message":"Patch Set 28:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-28-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"7b3a418a71e898416d78d8ef8e143e3efccfab90","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-01-09 12:13:27.000000000","message":"Patch Set 28:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/28/check/Dell_SC_FC/5a16e86/Dell_SC_FC/621 : SUCCESS in 1h 03m 25s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/28/check/Dell_SC_iSCSI/c616d57/Dell_SC_iSCSI/610 : SUCCESS in 1h 24m 56s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":28},{"id":"266daa2720edfe3949eef32a94b22e715ad6f586","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-01-09 12:14:41.000000000","message":"Patch Set 28:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/28/HPMSA-FC : SUCCESS in 34:06.26\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/28/HPMSA-iSCSI : SUCCESS in 34:18.67","accounts_in_message":[],"_revision_number":28},{"id":"760e654ad5062b3b7956241b76b2e9c8888fe76f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-09 12:17:38.000000000","message":"Patch Set 28:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/28/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f36f7c3 : FAILURE in 5m 23s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/28/check/3par-fc-driver-master-client-pip-ssa02-dsvm/63b3aa9 : FAILURE in 2m 27s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/28/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/4d79d3a : FAILURE in 26s","accounts_in_message":[],"_revision_number":28},{"id":"69894ccd962f9cb9a3e7d20ed481bb7afb68a976","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-01-09 12:18:50.000000000","message":"Patch Set 28:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/28/Lenovo-FC : SUCCESS in 35:53.99\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/28/Lenovo-iSCSI : SUCCESS in 35:20.81","accounts_in_message":[],"_revision_number":28},{"id":"3f8e9d9c49fe7337dbbd53a0b0bddb4ab81f40c2","author":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"date":"2019-01-09 13:52:12.000000000","message":"Patch Set 28:\n\nrecheck intel_rsd_ci","accounts_in_message":[],"_revision_number":28},{"id":"9f76effb8a4ff58240a049be1a304f9e52967664","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-09 14:46:11.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/28/iscsi/ : SUCCESS in 53m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/28/smb/ : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":28},{"id":"7bd992bf7b7d8e27ef321e55c0b5ff3ae156f341","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-09 15:03:46.000000000","message":"Patch Set 28:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/28/check/tempest-dsvm-full-xenial/d78c7db/ : SUCCESS in 1h 04m 40s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/28/check/devstack-plugin-ceph-dsvm-xenial/6457df7/ : SUCCESS in 1h 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"5cf2075802a8b767233cf3f8c94e6ad61a71c4c6","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-09 15:16:06.000000000","message":"Patch Set 28:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/28 : FAILURE","accounts_in_message":[],"_revision_number":28},{"id":"b6c6fc873ede0bb0010319971a9d57c2105268ab","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-09 15:23:54.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/28/check/dsvm-tempest-cisco-zonemanager-job/20fa395/ : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":28},{"id":"214b6c29b4606fcbdc1fb37ab7ab69b2b797d89d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-01-09 15:32:58.000000000","message":"Patch Set 28:\n\nrerun intel_rsd_ci","accounts_in_message":[],"_revision_number":28},{"id":"8692f13cb8eed020842dc738a9cadc9a6aca2c45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 15:52:19.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/28/check/openstack-tox-lower-constraints/f972539/ : SUCCESS in 12m 04s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/28/check/openstack-tox-pep8/cc3e4ff/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/28/check/openstack-tox-py27/7a05c4f/ : SUCCESS in 8m 18s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/28/check/openstack-tox-py35/8796397/ : SUCCESS in 9m 17s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/28/check/openstack-tox-py36/4aad52c/ : SUCCESS in 9m 35s\n- openstack-tox-docs http://logs.openstack.org/65/621465/28/check/openstack-tox-docs/d160910/html/ : SUCCESS in 11m 03s\n- requirements-check http://logs.openstack.org/65/621465/28/check/requirements-check/d2257c9/ : SUCCESS in 3m 11s\n- tempest-full http://logs.openstack.org/65/621465/28/check/tempest-full/d0afa95/ : SUCCESS in 1h 42m 06s\n- neutron-grenade http://logs.openstack.org/65/621465/28/check/neutron-grenade/0f656e9/ : SUCCESS in 1h 05m 39s\n- grenade-py3 http://logs.openstack.org/65/621465/28/check/grenade-py3/3a022c6/ : SUCCESS in 57m 57s\n- tempest-full-py3 http://logs.openstack.org/65/621465/28/check/tempest-full-py3/4c591e0/ : SUCCESS in 1h 13m 16s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/28/check/build-openstack-releasenotes/9f62941/html/ : SUCCESS in 12m 09s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/28/check/cinder-tempest-dsvm-lvm-lio-barbican/2542d7e/ : SUCCESS in 1h 41m 10s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/28/check/cinder-tox-bandit-baseline/2cf6177/ : SUCCESS in 7m 06s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/28/check/nova-multiattach/51897a9/ : FAILURE in 1h 09m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/28/check/openstack-tox-functional-py35/ed2e6c4/ : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/28/check/openstack-tox-functional-py36/49c3e59/ : SUCCESS in 6m 21s\n- cinder-rally-task http://logs.openstack.org/65/621465/28/check/cinder-rally-task/86378c3/ : SUCCESS in 42m 47s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/28/check/openstack-tox-pylint/f0b9f95/ : FAILURE in 5m 12s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/28/check/devstack-plugin-ceph-tempest/acc2ac3/ : SUCCESS in 1h 36m 19s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/28/check/openstack-tox-functional/ccf3870/ : SUCCESS in 4m 58s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/28/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/2375000/ : SUCCESS in 1h 07m 10s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-lvm-multibackend/56ad178/ : SUCCESS in 1h 10m 42s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-full-drbd-devstack/59f9047/ : SUCCESS in 1h 32m 53s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/3fa3af1/ : SUCCESS in 1h 16m 51s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/28/check/tempest-slow/73c8814/ : SUCCESS in 1h 53m 28s","accounts_in_message":[],"_revision_number":28},{"id":"9b9797df35eb177def13829de0ce86cd1d649741","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-09 16:06:28.000000000","message":"Patch Set 28:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/639 : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"0a5b5a4313403b00366a3b11f7e573f1c2e9f866","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-09 18:34:30.000000000","message":"Patch Set 28:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":28},{"id":"5036a9b2a6cb30906a9a9411222b9709409d8724","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2019-01-09 20:36:08.000000000","message":"Patch Set 28: Code-Review+1\n\nrecheck nova-multiattach\n\nLooks good to me, I\u0027m not sure why \"tempest.api.compute.servers.test_device_tagging.TaggedAttachmentsTest.test_tagged_attachment\" test is falling.","accounts_in_message":[],"_revision_number":28},{"id":"683bb07e6ed784aa1a2b87fe66011a92d15bdb52","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-09 20:36:30.000000000","message":"Patch Set 28:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":28},{"id":"d63575ce006badd5d11631a3db712a520e28031c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-09 21:27:12.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/28/iscsi/ : SUCCESS in 50m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/28/smb/ : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":28},{"id":"de747161829ebb8f76b077d47e3d505845b306a7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-09 21:48:07.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/28/check/dsvm-tempest-cisco-zonemanager-job/b1f652a/ : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":28},{"id":"80c1702e503e7f373b0865b67f61760937b09c80","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-09 21:56:24.000000000","message":"Patch Set 28:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/28/check/tempest-dsvm-full-xenial/5a0de8e/ : SUCCESS in 1h 19m 56s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/28/check/devstack-plugin-ceph-dsvm-xenial/9bcec9d/ : SUCCESS in 1h 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"4be2ad2653fdd29b6ea219857a5c23f67ea804ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 23:10:28.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/28/check/openstack-tox-lower-constraints/813f6d5/ : SUCCESS in 14m 50s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/28/check/openstack-tox-pep8/8a4e40f/ : SUCCESS in 11m 33s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/28/check/openstack-tox-py27/81eadf9/ : SUCCESS in 13m 21s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/28/check/openstack-tox-py35/66491c9/ : SUCCESS in 13m 17s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/28/check/openstack-tox-py36/06aedcf/ : SUCCESS in 13m 59s\n- openstack-tox-docs http://logs.openstack.org/65/621465/28/check/openstack-tox-docs/961b72e/html/ : SUCCESS in 11m 08s\n- requirements-check http://logs.openstack.org/65/621465/28/check/requirements-check/5bff18a/ : SUCCESS in 5m 25s\n- tempest-full http://logs.openstack.org/65/621465/28/check/tempest-full/bb05cbf/ : SUCCESS in 1h 44m 53s\n- neutron-grenade http://logs.openstack.org/65/621465/28/check/neutron-grenade/89ee2e1/ : SUCCESS in 1h 00m 40s\n- grenade-py3 http://logs.openstack.org/65/621465/28/check/grenade-py3/44902bb/ : SUCCESS in 1h 04m 16s\n- tempest-full-py3 http://logs.openstack.org/65/621465/28/check/tempest-full-py3/94fae65/ : SUCCESS in 1h 20m 57s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/28/check/build-openstack-releasenotes/e7fd1d3/html/ : SUCCESS in 11m 38s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/28/check/cinder-tempest-dsvm-lvm-lio-barbican/80ecad8/ : SUCCESS in 2h 26m 00s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/28/check/cinder-tox-bandit-baseline/3c46160/ : SUCCESS in 11m 03s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/28/check/nova-multiattach/9b40cbb/ : SUCCESS in 1h 10m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/28/check/openstack-tox-functional-py35/3f7313d/ : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/28/check/openstack-tox-functional-py36/b60b0bc/ : SUCCESS in 8m 46s\n- cinder-rally-task http://logs.openstack.org/65/621465/28/check/cinder-rally-task/8f38e3d/ : SUCCESS in 43m 19s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/28/check/openstack-tox-pylint/7080300/ : FAILURE in 8m 51s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/28/check/devstack-plugin-ceph-tempest/4c02c3c/ : SUCCESS in 1h 38m 20s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/28/check/openstack-tox-functional/ceca0bd/ : SUCCESS in 7m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/28/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/db5315e/ : SUCCESS in 1h 06m 35s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-lvm-multibackend/6d1b054/ : SUCCESS in 1h 30m 55s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-full-drbd-devstack/738d0e4/ : SUCCESS in 1h 32m 52s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/28/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/41f1c3a/ : SUCCESS in 1h 39m 51s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/28/check/tempest-slow/fcae6fe/ : SUCCESS in 2h 05m 09s","accounts_in_message":[],"_revision_number":28},{"id":"1e0e7c7ce40cd87342046152931f3025a51dee18","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-09 23:18:46.000000000","message":"Patch Set 28:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/28/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c5e0fe9 : FAILURE in 5m 44s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/28/check/3par-fc-driver-master-client-pip-ssa02-dsvm/b0950e8 : FAILURE in 2m 22s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/28/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/f1c5b12 : FAILURE in 23s","accounts_in_message":[],"_revision_number":28},{"id":"8823c79f9e66a4e1d226a0dd90d688bba161dc65","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-09 23:58:35.000000000","message":"Patch Set 28:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-28 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":28},{"id":"ab51b42982616e1936b148a9481a75379da83da7","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-01-09 23:59:57.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/28/check/synology-dsm-cinder-iscsi/e2b4ef5 : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":28},{"id":"ebe625d1214e7c05774b6fe691fe9df3c86d0d7e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-01-10 00:06:21.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/28/upstream-solidfire-check/cinder-solidfire-iSCSI/5ee1e41/ : FAILURE in 54m 33s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":28},{"id":"0b726098d57cddcfd4e8cf0c88bdcd095afd9376","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-01-10 00:25:48.000000000","message":"Patch Set 28:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-28-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"b3fc6ef6ec3efebdf5277c693e82e85288a5d857","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-01-10 00:29:57.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/65/621465/28/upstream-eseries-check/cinder-eseries-iSCSI/0615ad9/ : FAILURE in 1h 18m 19s (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/65/621465/28/upstream-eseries-check/cinder-eseries-FCP/f5d4f3f/ : FAILURE in 52m 49s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":28},{"id":"aa3da5845777fbb881fd72d9b068fbaa2d724b06","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-01-10 00:42:22.000000000","message":"Patch Set 28:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/28/check/Dell_SC_FC/a5b8df1/Dell_SC_FC/624 : SUCCESS in 1h 06m 22s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/28/check/Dell_SC_iSCSI/f142193/Dell_SC_iSCSI/613 : SUCCESS in 1h 31m 30s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":28},{"id":"a41b73d65d353fbcedbe0efc08217c001f31a2f7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-10 00:49:31.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/28/upstream-check/cinder-cDOT-iSCSI/98e61a0/ : FAILURE in 45m 11s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/28/upstream-check/cinder-cDOT-NFS/c311223/ : FAILURE in 1h 38m 31s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/28/upstream-check/cinder-cDOT-FCP/a5942a4/ : FAILURE in 1h 04m 32s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":28},{"id":"a7c435c69c5640dc528848d6f6171eb28e8b27ed","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-01-10 00:54:03.000000000","message":"Patch Set 28:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/28/Lenovo-FC : SUCCESS in 33:17.91\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/28/Lenovo-iSCSI : SUCCESS in 32:56.45","accounts_in_message":[],"_revision_number":28},{"id":"8be77e74caa46c2f723234a8b229ecd3a45150d4","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-01-10 00:56:46.000000000","message":"Patch Set 28:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/28/HPMSA-FC : SUCCESS in 34:02.82\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/28/HPMSA-iSCSI : SUCCESS in 34:01.37","accounts_in_message":[],"_revision_number":28},{"id":"8f8d212dc1a945d45ac7e02b39e9f32cf320e35f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-10 01:08:19.000000000","message":"Patch Set 28:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/28 : FAILURE","accounts_in_message":[],"_revision_number":28},{"id":"7c6baf0e4856e35e556eb3921c679fbc458d11d8","author":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"date":"2019-01-10 10:49:35.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"add93e1654cba7bee3c2f6dfea80e153fa6160e9","author":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"date":"2019-01-10 11:22:12.000000000","message":"Patch Set 28:\n\nrun-nexenta edge ci","accounts_in_message":[],"_revision_number":28},{"id":"87eabe92d4d79603e7ebe6bb0f355f270a3cf16e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-01-10 17:40:43.000000000","message":"Patch Set 28:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":28},{"id":"6f41f89c4966c7f947be9c9ba4f0e5d8f519067f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-10 17:46:43.000000000","message":"Patch Set 28:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/28 : FAILURE","accounts_in_message":[],"_revision_number":28},{"id":"a33621ceeffe7d20dda27b70c65f10916557d92f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-01-10 17:53:14.000000000","message":"Patch Set 28: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":28},{"id":"1156a5632862b4374b9251306b16526d7a42c3e6","author":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"date":"2019-01-10 20:48:25.000000000","message":"Patch Set 28:\n\n(10 comments)","accounts_in_message":[],"_revision_number":28},{"id":"11a8fe7100c700f41f1aeb57afefa482f5eb075e","author":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"date":"2019-01-10 21:13:51.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"d2e42ecd3c6a3f27dbb3c695a3c9bcae37cda654","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-10 21:14:42.000000000","message":"Patch Set 29:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/29/check/dsvm-veritas-access-cinder-driver/703aff7 : FAILURE in 12s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/29/check/dsvm-veritas-access-cinder-iscsi-driver/7f52071 : FAILURE in 12s","accounts_in_message":[],"_revision_number":29},{"id":"02a9fcd78ae3b298c3b354bb326dbcfc51156185","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-01-10 21:16:49.000000000","message":"Patch Set 29: Code-Review+1\n\nLooks like all comments were addressed. Thanks for the quick response!\n\nJust waiting for CI results now.","accounts_in_message":[],"_revision_number":29},{"id":"1c5fa38fa4da2ebd1bcfdecec2e85424fe479423","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-10 21:24:31.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Unable to find playbook /tmp/tmp9_yj7755/1f5a38aad1b74e04896abae43379e0de/trusted/project_0/git.zuul-ci.org/zuul-base-jobs/playbooks/base/post-ssh.yaml","accounts_in_message":[],"_revision_number":29},{"id":"41e4377ca017c6f71729188bca66d9a6f668fe01","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-01-10 21:30:50.000000000","message":"Patch Set 29:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/546 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/743 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/912 : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"b18b291e998bc872db58df529a76da40f94d0250","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-10 21:50:49.000000000","message":"Patch Set 29:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/29 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"4bb37a08e56736109d4f17af21a98fcc1c222cde","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-10 22:05:28.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/29/iscsi/ : SUCCESS in 51m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/29/smb/ : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":29},{"id":"eb34f75609dc56d29c57b853e1b11c1257416069","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-10 22:15:32.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/29/check-cinder/Cinder-ISER-ISCSI/7565089 : SUCCESS in 58m 53s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/29/check-cinder/Cinder-ISER-LIO/0471654 : SUCCESS in 1h 00m 42s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":29},{"id":"7f5c25e82c09046746598b24d2cf0b9d65a01fdc","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-10 22:18:01.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/29/check/dsvm-tempest-kvm/ed88f5b : SUCCESS in 1h 03m 30s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/29/check/dsvm-tempest-exe/1c2e3f7 : SUCCESS in 53m 41s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":29},{"id":"d80899329f823a1d7b3dcdcaebe49fdbe367ab6f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-10 22:18:26.000000000","message":"Patch Set 29:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/29/check-not-vote/ext-cinder-zuul/2ffe947 : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":29},{"id":"906685192e17a805e37f99eab25fa405be07ae30","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-10 22:19:49.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/29/check/dsvm-tempest-infinibox-fc/5b036c2 : SUCCESS in 51m 00s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/29/check/dsvm-tempest-infinibox-iscsi/953350e : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":29},{"id":"03b376354ab853e3915092bca81ca0165f059372","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-10 22:24:19.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/29/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1f9065d : SUCCESS in 44m 30s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/29/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/22dbea1 : SUCCESS in 1h 10m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":29},{"id":"867afdab18539bd3f1770535ff7e20c1a964d28d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-10 22:26:05.000000000","message":"Patch Set 29:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/29/check/tempest-dsvm-full-xenial/165369a/ : SUCCESS in 1h 11m 56s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/29/check/devstack-plugin-ceph-dsvm-xenial/17c83bb/ : SUCCESS in 1h 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"898bdf99797553cc25f9a10098f51d66a8a4e488","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-10 22:34:02.000000000","message":"Patch Set 29:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/29 : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"27fa322156bad7318d729b7a9186338110a51bf4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-10 22:38:27.000000000","message":"Patch Set 29:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/29/check/fujitsu-eternus-dx-iscsi/a0b86da : SUCCESS in 1h 23m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/29/check/fujitsu-eternus-dx-fc/f0edc68 : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":29},{"id":"220251af029f305a7c0bab95759f7c574261686d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-10 23:18:23.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/29/check/openstack-tox-lower-constraints/06a5246/ : SUCCESS in 10m 21s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/29/check/openstack-tox-pep8/fca89fd/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/29/check/openstack-tox-py27/042b234/ : SUCCESS in 8m 47s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/29/check/openstack-tox-py35/f098071/ : SUCCESS in 8m 38s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/29/check/openstack-tox-py36/bebb7b3/ : SUCCESS in 9m 50s\n- openstack-tox-docs http://logs.openstack.org/65/621465/29/check/openstack-tox-docs/293ae85/html/ : SUCCESS in 9m 22s\n- requirements-check http://logs.openstack.org/65/621465/29/check/requirements-check/f25e01b/ : SUCCESS in 3m 37s\n- tempest-full http://logs.openstack.org/65/621465/29/check/tempest-full/73dc554/ : SUCCESS in 1h 30m 29s\n- neutron-grenade http://logs.openstack.org/65/621465/29/check/neutron-grenade/ef8fdce/ : SUCCESS in 50m 26s\n- grenade-py3 http://logs.openstack.org/65/621465/29/check/grenade-py3/e4e0669/ : SUCCESS in 1h 06m 40s\n- tempest-full-py3 http://logs.openstack.org/65/621465/29/check/tempest-full-py3/c7b8db7/ : SUCCESS in 1h 50m 59s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/29/check/build-openstack-releasenotes/2614771/html/ : SUCCESS in 12m 29s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/29/check/cinder-tempest-dsvm-lvm-lio-barbican/3367338/ : SUCCESS in 1h 56m 12s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/29/check/cinder-tox-bandit-baseline/8cf788e/ : SUCCESS in 7m 25s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/29/check/nova-multiattach/ca7917e/ : SUCCESS in 53m 33s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/29/check/openstack-tox-functional-py35/9d8968d/ : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/29/check/openstack-tox-functional-py36/55666cd/ : SUCCESS in 4m 55s\n- cinder-rally-task http://logs.openstack.org/65/621465/29/check/cinder-rally-task/8b035ef/ : SUCCESS in 34m 48s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/29/check/openstack-tox-pylint/d2dbb74/ : SUCCESS in 4m 46s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/29/check/devstack-plugin-ceph-tempest/9bd4927/ : SUCCESS in 1h 44m 22s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/29/check/openstack-tox-functional/a61171b/ : SUCCESS in 4m 52s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/29/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/093bfb3/ : SUCCESS in 58m 58s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/29/check/legacy-tempest-dsvm-lvm-multibackend/49f8e1c/ : SUCCESS in 1h 04m 46s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/29/check/legacy-tempest-dsvm-full-drbd-devstack/79b54a1/ : SUCCESS in 1h 27m 09s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/29/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/5a4e67d/ : SUCCESS in 1h 28m 30s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/29/check/tempest-slow/008cafc/ : SUCCESS in 1h 44m 40s","accounts_in_message":[],"_revision_number":29},{"id":"ded95371ee99e7b209833dfee5c86bdb98fa2189","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-10 23:26:56.000000000","message":"Patch Set 29:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/29/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4a6a316 : FAILURE in 5m 58s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/29/check/3par-fc-driver-master-client-pip-ssa02-dsvm/ff47e85 : FAILURE in 2m 48s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/29/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/504edd1 : FAILURE in 51s","accounts_in_message":[],"_revision_number":29},{"id":"5343b1de544abcc0fa482ad84b21628c80cbf664","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-01-10 23:57:46.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/65/621465/29/upstream-eseries-check/cinder-eseries-iSCSI/218c256/ : FAILURE in 38m 24s (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/65/621465/29/upstream-eseries-check/cinder-eseries-FCP/d4669bd/ : FAILURE in 33m 17s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":29},{"id":"6d4a6d4234203854b52c85cce635132ef869a516","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-01-11 00:02:25.000000000","message":"Patch Set 29:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-29/nedge_iscsi : FAILURE in 41m 46s","accounts_in_message":[],"_revision_number":29},{"id":"6e595683daafb86fbfc3e29deea4b123895daf9f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-11 00:06:49.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/29/upstream-check/cinder-cDOT-iSCSI/3ad6dab/ : FAILURE in 30m 01s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/29/upstream-check/cinder-cDOT-NFS/27c2656/ : FAILURE in 30m 18s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/29/upstream-check/cinder-cDOT-FCP/bea3f02/ : FAILURE in 47m 49s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"ab1a54d9a1b521804696ef784a67be251bd7a620","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-11 00:07:10.000000000","message":"Patch Set 29:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-29 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":29},{"id":"804a4a98ef597e842c30726bcba54c3534f12b27","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-01-11 00:10:13.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/29/check/synology-dsm-cinder-iscsi/30053b2 : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":29},{"id":"114537788015e503c3f7c7c973fc24ea12bc33a4","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-01-11 00:51:23.000000000","message":"Patch Set 29:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-29-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"9e25c310e321b957a3a5fcdbf1c011654e7c80a0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-01-11 01:07:01.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/29/upstream-solidfire-check/cinder-solidfire-iSCSI/89fc7fe/ : FAILURE in 1h 47m 36s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":29},{"id":"5a78a56379080ccf72ba810ee75a642b5f935bbb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-01-11 01:33:25.000000000","message":"Patch Set 29:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/29/check/Dell_SC_FC/fc904ee/Dell_SC_FC/635 : SUCCESS in 1h 09m 57s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/29/check/Dell_SC_iSCSI/09821d6/Dell_SC_iSCSI/624 : SUCCESS in 1h 38m 52s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":29},{"id":"a43e8dc5116b14a200567d08a967d54b41a6bbe0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-01-11 01:56:04.000000000","message":"Patch Set 29:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/29/HPMSA-FC : SUCCESS in 33:45.40\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/29/HPMSA-iSCSI : SUCCESS in 35:10.53","accounts_in_message":[],"_revision_number":29},{"id":"24182f164fde954d3da46c988a7b2c7524974490","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-01-11 01:56:26.000000000","message":"Patch Set 29:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/29/Lenovo-FC : SUCCESS in 35:03.20\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/29/Lenovo-iSCSI : SUCCESS in 35:13.44","accounts_in_message":[],"_revision_number":29},{"id":"8c9cf837e56e04a2829bdcae44b8417fc2856247","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-11 03:12:10.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/29/check/Cinder-AS13000/187bf19 : SUCCESS in 2h 29m 31s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/29/check/cinder-test-fc/d64e7ea : SUCCESS in 56m 55s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/29/check/cinder-test-iscsi/7f49c9d : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":29},{"id":"5c9bbe28445369d70e5c70196fe90fe62f89c582","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-11 05:47:38.000000000","message":"Patch Set 29:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/654 : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"11e41a9b65faa824c83e8db30040b72815f19174","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-11 06:45:54.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/29/check/NEC-ISCSI/89e2f06 : SUCCESS in 1h 52m 28s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/29/check/NEC-FC/bca9cfb : SUCCESS in 1h 53m 59s","accounts_in_message":[],"_revision_number":29},{"id":"b5b5d8e1f38deacdda91c8ca476798ae0a621658","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-11 08:07:26.000000000","message":"Patch Set 29:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/656 : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"88d4e28da66193a3fd5f3f2dd20d69ca412283fa","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-01-11 12:02:28.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VNX_iSCSI http://publiclogs.emc.com/621465/29/EMC_VNX_iSCSI/365 : FAILURE in 3h 54m 06s\n- EMC_VNX_FC http://publiclogs.emc.com/621465/29/EMC_VNX_FC/340 : FAILURE in 2h 09m 36s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":29},{"id":"f8d9d99046759c8dacaeba013a787896d5c25dd8","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-11 13:43:43.000000000","message":"Patch Set 29:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/29 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"a144a5319a93c7ad7a0ba84d910f5ecf9035c4e7","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-11 17:00:47.000000000","message":"Patch Set 29:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/29 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"652054d617333e5cef514547a77b45568cce1eda","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-11 20:57:50.000000000","message":"Patch Set 29:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/29 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"0469efd5274bc29d7e3862c047b38ae8a0db344f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-12 05:26:12.000000000","message":"Patch Set 29:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/29 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"6e2c76a86c662841e8f25461a086c76b319edc63","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-01-13 06:55:18.000000000","message":"Patch Set 29:\n\nBuild succeed.\n\n- EMC_VMAX_FC http://publiclogs.emc.com/621465/29/EMC_VMAX_FC/769 : SUCCESS in 6h 58m 54s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/621465/29/EMC_VMAX_iSCSI/756 : SUCCESS in 6h 58m 43s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":29},{"id":"968f4e33230c53cc677c2a4ed88e3ac104d7605e","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-01-14 18:09:44.000000000","message":"Patch Set 29: Code-Review-1\n\nSince this is a new driver, you might want to put the @utils.trace decorators on all the main entry points of the driver to help debugging issues in the field.","accounts_in_message":[],"_revision_number":29},{"id":"82c05632da81d14cb90c7a7ade5a65e65ab805b3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-01-14 19:49:34.000000000","message":"Patch Set 29: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"86b0b6af16ca1ded002bc3633bb330610ebbbd51","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-15 01:32:08.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"3cf13e3523d377ba7e5250fa81c150515c297ca6","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-15 01:33:29.000000000","message":"Patch Set 30:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/30/check/dsvm-veritas-access-cinder-driver/6fe445d : FAILURE in 30s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/30/check/dsvm-veritas-access-cinder-iscsi-driver/8b211c8 : FAILURE in 11s","accounts_in_message":[],"_revision_number":30},{"id":"2e745c0b09c4d3acfac34a788bf8840f6d9f4966","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-15 01:39:39.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"e0145e990364456d48afbf022cb0bc09a524802f","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-15 01:40:29.000000000","message":"Patch Set 31:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/31/check/dsvm-veritas-access-cinder-driver/e51f69d : FAILURE in 11s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/31/check/dsvm-veritas-access-cinder-iscsi-driver/0dfd12a : FAILURE in 11s","accounts_in_message":[],"_revision_number":31},{"id":"3ccf04195803c3e7c46ef4c1bd78ed08714fcefe","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-15 01:47:53.000000000","message":"Patch Set 31:\n\n(1 comment)\n\n@Walter: Addressed, thanks!","accounts_in_message":[],"_revision_number":31},{"id":"ea7cd0cb169a46ccc7c51fbb9396917858c261c9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-15 02:02:23.000000000","message":"Patch Set 31:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/31/iscsi/ : FAILURE in 18m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/31/smb/ : FAILURE in 18m 23s","accounts_in_message":[],"_revision_number":31},{"id":"de72fbe06876f0e8fc6a42e259080739da8be4c0","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-01-15 02:11:00.000000000","message":"Patch Set 30:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/618 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/781 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/950 : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"913d640b34acb46af2bbc24d52e7c6d065914a20","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-01-15 02:19:35.000000000","message":"Patch Set 31:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/619 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"6a2083b755df3ce87c843b6ee5387197538f435e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-15 02:45:30.000000000","message":"Patch Set 31:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/31/check/fujitsu-eternus-dx-iscsi/3fe538e : FAILURE in 9m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/31/check/fujitsu-eternus-dx-fc/4055152 : FAILURE in 11m 14s","accounts_in_message":[],"_revision_number":31},{"id":"f6f284a1adb4cf02193e53701474149bb37a36f1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-15 02:50:19.000000000","message":"Patch Set 31:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/31/check/tempest-dsvm-full-xenial/49cba47/ : SUCCESS in 1h 10m 14s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/31/check/devstack-plugin-ceph-dsvm-xenial/f70a1d5/ : FAILURE in 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"22dc4da95459138db3994f01044ac94bf8298907","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-15 04:06:13.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/31/check/openstack-tox-lower-constraints/a53f086/ : SUCCESS in 19m 39s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/31/check/openstack-tox-pep8/1f9fc81/ : SUCCESS in 7m 39s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/31/check/openstack-tox-py27/54fc0e4/ : SUCCESS in 9m 07s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/31/check/openstack-tox-py35/87f2b46/ : SUCCESS in 8m 24s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/31/check/openstack-tox-py36/083f000/ : SUCCESS in 8m 28s\n- openstack-tox-docs http://logs.openstack.org/65/621465/31/check/openstack-tox-docs/b7289b3/html/ : SUCCESS in 9m 37s\n- requirements-check http://logs.openstack.org/65/621465/31/check/requirements-check/44a7bb1/ : SUCCESS in 2m 32s\n- tempest-full http://logs.openstack.org/65/621465/31/check/tempest-full/189f8f3/ : SUCCESS in 1h 40m 33s\n- neutron-grenade http://logs.openstack.org/65/621465/31/check/neutron-grenade/6a9641e/ : SUCCESS in 54m 43s\n- grenade-py3 http://logs.openstack.org/65/621465/31/check/grenade-py3/7e4e47b/ : SUCCESS in 54m 04s\n- tempest-full-py3 http://logs.openstack.org/65/621465/31/check/tempest-full-py3/a8978f5/ : SUCCESS in 1h 18m 47s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/31/check/build-openstack-releasenotes/e1a0801/html/ : SUCCESS in 13m 15s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/31/check/cinder-tempest-dsvm-lvm-lio-barbican/93ece3e/ : SUCCESS in 2h 15m 29s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/31/check/cinder-tox-bandit-baseline/1411b2a/ : SUCCESS in 7m 37s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/31/check/nova-multiattach/66fbc6c/ : SUCCESS in 1h 02m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/31/check/openstack-tox-functional-py35/fb13810/ : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/31/check/openstack-tox-functional-py36/070da49/ : SUCCESS in 4m 31s\n- cinder-rally-task http://logs.openstack.org/65/621465/31/check/cinder-rally-task/c740aaf/ : SUCCESS in 52m 16s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/31/check/openstack-tox-pylint/0eb4c6a/ : SUCCESS in 4m 38s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/31/check/devstack-plugin-ceph-tempest/069cfa1/ : SUCCESS in 1h 41m 58s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/31/check/openstack-tox-functional/20df4bb/ : SUCCESS in 4m 43s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/31/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/f7ef65f/ : SUCCESS in 1h 06m 46s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/31/check/legacy-tempest-dsvm-lvm-multibackend/bf57449/ : SUCCESS in 1h 17m 25s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/31/check/legacy-tempest-dsvm-full-drbd-devstack/cc76e69/ : SUCCESS in 1h 10m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/31/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/b87b18e/ : SUCCESS in 1h 16m 25s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/31/check/tempest-slow/6dcf150/ : SUCCESS in 1h 32m 06s","accounts_in_message":[],"_revision_number":31},{"id":"392e6e9eb462cf445d289f3482f168707057abba","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-15 04:06:27.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/31/check/dsvm-tempest-kvm/460bd54 : FAILURE in 30m 51s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/31/check/dsvm-tempest-exe/37a50e2 : FAILURE in 45m 49s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":31},{"id":"64fb2e65569d0f5a263adbac0ee5ce7b13bb49c6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-15 04:46:50.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/31/check/dsvm-tempest-infinibox-fc/62d144e : SUCCESS in 1h 03m 09s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/31/check/dsvm-tempest-infinibox-iscsi/35ea314 : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":31},{"id":"d471a045f4ad099625715e325bfa72e29ff02232","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-01-15 04:52:01.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/31/upstream-solidfire-check/cinder-solidfire-iSCSI/3cf285b/ : FAILURE in 44m 15s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":31},{"id":"aa6dfe38d3d784906bb72e5a27be7f0e3c9c3e9d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-15 04:57:33.000000000","message":"Patch Set 30:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/30 : FAILURE","accounts_in_message":[],"_revision_number":30},{"id":"b7d5201ceea33898032fade56c97cd890176bd36","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-01-15 05:33:38.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/65/621465/31/upstream-eseries-check/cinder-eseries-iSCSI/15d4b68/ : POST_FAILURE in 17m 20s (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/65/621465/31/upstream-eseries-check/cinder-eseries-FCP/6c193e3/ : FAILURE in 38m 04s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":31},{"id":"ae15f9c07d3ea48b5192f4d9a2a6fb61c93b802a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-15 05:46:12.000000000","message":"Patch Set 30:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/30 : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"27590c2da1794a8ef7ffc58401a6312861dff5a2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-15 06:08:43.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/31/upstream-check/cinder-cDOT-iSCSI/4507fbb/ : FAILURE in 30m 32s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/31/upstream-check/cinder-cDOT-NFS/37f9b5d/ : FAILURE in 1h 16m 52s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/31/upstream-check/cinder-cDOT-FCP/de16916/ : FAILURE in 54m 47s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":31},{"id":"46a247a536f11e8429690cfc8017e51716aec3e0","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-01-15 06:15:54.000000000","message":"Patch Set 31:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-31/nedge_iscsi : SUCCESS in 25m 29s","accounts_in_message":[],"_revision_number":31},{"id":"9b9d12ebcd6047c32b17b5061ef8458b549ea757","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-15 06:52:37.000000000","message":"Patch Set 31:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/31 : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"4d1a5be0a3d3dfdc767252389c582749d5fbc8aa","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-15 08:22:30.000000000","message":"Patch Set 31:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/31 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"8150a01574e1addff3c62d0ae1146fb4a4a16e38","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-01-15 08:39:16.000000000","message":"Patch Set 31:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/31/check/EMC_Unity_FC/02f89e1/EMC_Unity_FC/753 : FAILURE in 44m 44s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/31/check/EMC_Unity_iSCSI/30a23b0/EMC_Unity_iSCSI/726 : FAILURE in 43m 14s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":31},{"id":"1b366052bcf91650a22e52f40375c14870a54815","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-15 10:12:21.000000000","message":"Patch Set 31:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-31 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":31},{"id":"32b2cf336c871daf69f1f4ce9695e48695f42ce0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-15 10:37:38.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/31/check/dsvm-tempest-cisco-zonemanager-job/19bccb2/ : FAILURE in 1h 01m 29s","accounts_in_message":[],"_revision_number":31},{"id":"607212ac355d26729b37abcea131bf6c47aea3c2","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-01-15 10:38:47.000000000","message":"Patch Set 31:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/31/check/synology-dsm-cinder-iscsi/a1eec1b : FAILURE in 55m 33s","accounts_in_message":[],"_revision_number":31},{"id":"0347ced11dab67550c23231e064309471699e4f5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-15 11:04:26.000000000","message":"Patch Set 31:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/31/check-not-vote/ext-cinder-zuul/5deec24 : SUCCESS in 40m 18s","accounts_in_message":[],"_revision_number":31},{"id":"86f31a8de5a8383db02787bb1e84609684243f74","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-15 12:38:20.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/31/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/9161a84 : SUCCESS in 45m 44s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/31/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/4f55aa1 : SUCCESS in 1h 09m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":31},{"id":"1e3b51e7e81d229d9a3209cc76a627cb304259c4","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-15 13:43:38.000000000","message":"Patch Set 31:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/31 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"7ff8e1256457b6b78343d186af1c9d3aaf777382","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-15 13:50:33.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/31/check/NEC-ISCSI/f108624 : SUCCESS in 1h 42m 10s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/31/check/NEC-FC/b9c4c28 : SUCCESS in 1h 53m 43s","accounts_in_message":[],"_revision_number":31},{"id":"5c05e8f30faeabbb65d2f798320e7b7bdbe5f7ac","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-01-15 13:54:07.000000000","message":"Patch Set 31:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-31-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"9691ad0effd9fb11d6b3bdc2018b79b7e00512d8","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-15 14:19:12.000000000","message":"Patch Set 30:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/754 : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"94a84e9c9f95355d5c765ddd87c39f49105669ab","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-15 15:15:56.000000000","message":"Patch Set 31:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/755 : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"62f67ac207c83892619cc4b48faba035136d8b33","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-15 17:02:49.000000000","message":"Patch Set 31:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/31 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"c159a55c26f8e06710690d30fac9754242360062","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-15 17:13:13.000000000","message":"Patch Set 31:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/31/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9c86ee6 : FAILURE in 48m 01s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/31/check/3par-fc-driver-master-client-pip-ssa02-dsvm/9a2930d : FAILURE in 1m 26s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/31/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/e0e95ea : FAILURE in 36m 24s","accounts_in_message":[],"_revision_number":31},{"id":"79fb1cb8142335a83f40dd1948fa388d61ad352b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-01-15 17:16:36.000000000","message":"Patch Set 31:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/31/check/Dell_SC_FC/e4382ba/Dell_SC_FC/677 : SUCCESS in 1h 00m 31s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/31/check/Dell_SC_iSCSI/3f82699/Dell_SC_iSCSI/666 : SUCCESS in 1h 28m 25s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":31},{"id":"e3f0578e002898d5cc6379d9a61d982751f96b2b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-01-15 17:29:34.000000000","message":"Patch Set 31: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"59175a111e6d413973d044bc16ee6bcf3903b8d4","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-01-15 21:15:19.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/31/check/Cinder-AS13000/f27055d : SUCCESS in 2h 21m 08s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/31/check/cinder-test-fc/2609669 : SUCCESS in 1h 06m 25s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/31/check/cinder-test-iscsi/de03cb7 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":31},{"id":"9ec6356efcdcd8b475f23e7ff6c90983c19e7cda","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-01-15 22:04:16.000000000","message":"Patch Set 31:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/31/HPMSA-FC : FAILURE in 9:40.62\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/31/HPMSA-iSCSI : FAILURE in 9:25.46","accounts_in_message":[],"_revision_number":31},{"id":"e23f74555d61b6253adb473506a487d7e4bd5bd3","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-15 22:49:27.000000000","message":"Patch Set 31:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/31 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"996a9284837214fc8c213c2992f8be3e626512c0","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-16 01:35:32.000000000","message":"Patch Set 31:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/31 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"43e8d33acfef1c0f21970171be17aa1fcda6c88b","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-16 12:08:12.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"aff1a8a0ad7dba2af701104cb37b3ad48515d05c","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-16 12:09:08.000000000","message":"Patch Set 32:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/32/check/dsvm-veritas-access-cinder-driver/902bec3 : FAILURE in 12s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/32/check/dsvm-veritas-access-cinder-iscsi-driver/8949228 : FAILURE in 12s","accounts_in_message":[],"_revision_number":32},{"id":"a487ca83117d17190472347fa7f0fa3d2cecfd29","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-16 12:10:57.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"d32b8676a169e4c9d2f8a5a192d7538a95739b34","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-01-16 12:25:07.000000000","message":"Patch Set 32:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/803 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/972 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/651 : FAILURE","accounts_in_message":[],"_revision_number":32},{"id":"465e488a82039af15766d54095982824499455f6","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-16 13:10:53.000000000","message":"Patch Set 31:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/31 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"d3d680f5457168bd029ed78e686d3fa43d7300bc","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-16 13:10:57.000000000","message":"Patch Set 32:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/32/check-cinder/Cinder-ISER-ISCSI/209c928 : FAILURE in 1h 00m 53s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/32/check-cinder/Cinder-ISER-LIO/7270f09 : FAILURE in 1h 01m 40s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":32},{"id":"a315035fbd82522d7ce36802aea6a24bb224dd62","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-16 13:24:20.000000000","message":"Patch Set 32:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/32/check/tempest-dsvm-full-xenial/26e08ab/ : SUCCESS in 1h 08m 39s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/32/check/devstack-plugin-ceph-dsvm-xenial/e3eca88/ : FAILURE in 1h 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"3f1a7929034b2d5dbae3d1004fbe812e4822e769","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-16 13:26:41.000000000","message":"Patch Set 32:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/32 : SUCCESS","accounts_in_message":[],"_revision_number":32},{"id":"ad514ce264fb2b8b920876eae055acf6b709a427","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-16 13:31:54.000000000","message":"Patch Set 32:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/32/check/fujitsu-eternus-dx-iscsi/aded128 : SUCCESS in 1h 21m 00s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/32/check/fujitsu-eternus-dx-fc/6ea68ee : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":32},{"id":"f408066034c1157b8b79e4cec72b2e103243af3e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-16 13:32:26.000000000","message":"Patch Set 32:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/32/iscsi/ : FAILURE in 1h 23m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/32/smb/ : SUCCESS in 48m 14s","accounts_in_message":[],"_revision_number":32},{"id":"b9932f46da9705646aa093a9a40c2dc40334e7f3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-16 13:34:48.000000000","message":"Patch Set 32:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/32/check/dsvm-tempest-infinibox-fc/f345082 : FAILURE in 1h 15m 41s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/32/check/dsvm-tempest-infinibox-iscsi/ff2a8be : SUCCESS in 1h 25m 41s","accounts_in_message":[],"_revision_number":32},{"id":"56efd38d37d094a537cb7f2b321aab6990d8714c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-16 14:20:31.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/32/check/openstack-tox-lower-constraints/f9cf9c5/ : SUCCESS in 9m 56s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/32/check/openstack-tox-pep8/62fe66d/ : SUCCESS in 7m 55s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/32/check/openstack-tox-py27/f5a691c/ : SUCCESS in 9m 05s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/32/check/openstack-tox-py35/542b85d/ : SUCCESS in 9m 56s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/32/check/openstack-tox-py36/3e3fb10/ : SUCCESS in 7m 58s\n- openstack-tox-docs http://logs.openstack.org/65/621465/32/check/openstack-tox-docs/347b8d2/html/ : SUCCESS in 9m 59s\n- requirements-check http://logs.openstack.org/65/621465/32/check/requirements-check/fce1ff8/ : SUCCESS in 3m 29s\n- tempest-full http://logs.openstack.org/65/621465/32/check/tempest-full/3960b70/ : SUCCESS in 1h 22m 27s\n- neutron-grenade http://logs.openstack.org/65/621465/32/check/neutron-grenade/22b8e46/ : SUCCESS in 1h 00m 02s\n- grenade-py3 http://logs.openstack.org/65/621465/32/check/grenade-py3/f3beea0/ : SUCCESS in 57m 43s\n- tempest-full-py3 http://logs.openstack.org/65/621465/32/check/tempest-full-py3/9c27c2b/ : SUCCESS in 1h 14m 09s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/32/check/build-openstack-releasenotes/0e9fbe5/html/ : SUCCESS in 13m 32s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/32/check/cinder-tempest-dsvm-lvm-lio-barbican/01e33ce/ : FAILURE in 2h 08m 41s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/32/check/cinder-tox-bandit-baseline/ed74f8b/ : SUCCESS in 7m 09s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/32/check/nova-multiattach/87fd8a4/ : SUCCESS in 1h 03m 47s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/32/check/openstack-tox-functional-py35/ea1dd12/ : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/32/check/openstack-tox-functional-py36/c281077/ : SUCCESS in 5m 28s\n- cinder-rally-task http://logs.openstack.org/65/621465/32/check/cinder-rally-task/fbf31f6/ : SUCCESS in 51m 40s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/32/check/openstack-tox-pylint/77cf003/ : SUCCESS in 6m 37s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/32/check/devstack-plugin-ceph-tempest/fcf9825/ : TIMED_OUT in 2h 06m 03s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/32/check/openstack-tox-functional/298be59/ : SUCCESS in 5m 41s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/32/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/1fa2c10/ : SUCCESS in 1h 09m 21s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/32/check/legacy-tempest-dsvm-lvm-multibackend/16df3c5/ : SUCCESS in 1h 19m 01s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/32/check/legacy-tempest-dsvm-full-drbd-devstack/0c80532/ : SUCCESS in 1h 18m 49s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/32/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/ec6ade8/ : SUCCESS in 2h 00m 41s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/32/check/tempest-slow/120c3eb/ : SUCCESS in 1h 32m 51s","accounts_in_message":[],"_revision_number":32},{"id":"890bf7e74f40f1e4f8fa2716da90c0fecdff5185","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-16 14:53:36.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/32/check/dsvm-tempest-kvm/a917cc3 : SUCCESS in 1h 12m 57s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/32/check/dsvm-tempest-exe/c71847c : SUCCESS in 58m 07s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":32},{"id":"a1a341c744134a877c79eafe05f18a54103800e6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-01-16 16:01:43.000000000","message":"Patch Set 32:\n\nWrote a tempest bug for the lio-barbican job failure here.\n    https://bugs.launchpad.net/tempest/+bug/1812036","accounts_in_message":[],"_revision_number":32},{"id":"429e81850c2afea52645693ff3855f6499edb14e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-01-16 16:01:49.000000000","message":"Patch Set 32:\n\nrecheck","accounts_in_message":[],"_revision_number":32},{"id":"b314f4a0685d2c7c01bad7ceaa51ecf8cdd158bf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-16 16:11:00.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/32/check/dsvm-tempest-cisco-zonemanager-job/7c83a66/ : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":32},{"id":"8947f05719808948319c6ceee697be35e1ec476e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-16 16:53:43.000000000","message":"Patch Set 32:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/32 : FAILURE","accounts_in_message":[],"_revision_number":32},{"id":"4502646aab45a4c8d8a91f845295c18debe6596c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-16 16:54:54.000000000","message":"Patch Set 32:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/32/iscsi/ : SUCCESS in 52m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/32/smb/ : FAILURE in 11m 38s","accounts_in_message":[],"_revision_number":32},{"id":"b7f4c9fd2bd2d39baf58d9162e0cdc516f4a2215","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-01-16 17:05:10.000000000","message":"Patch Set 31:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/31/Lenovo-FC : FAILURE in 4:25:42\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/31/Lenovo-iSCSI : FAILURE in 6:08:23","accounts_in_message":[],"_revision_number":31},{"id":"f653f40dc6bb0d24e97b9a2aca4fe276275b81c8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-16 17:20:42.000000000","message":"Patch Set 32:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/32/check/tempest-dsvm-full-xenial/acaa1d2/ : SUCCESS in 1h 16m 58s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/32/check/devstack-plugin-ceph-dsvm-xenial/ef971d0/ : SUCCESS in 1h 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"d1317b144e82fe1b0a325a2ee6fb07b07d016d26","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-16 18:40:26.000000000","message":"Patch Set 32:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/32/check/NEC-ISCSI/e16f40c : FAILURE in 1h 39m 23s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/32/check/NEC-FC/7ed56e7 : FAILURE in 58m 21s","accounts_in_message":[],"_revision_number":32},{"id":"69b7d681893512378929c428f1b79afb8b9ea78d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-16 18:55:17.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/32/check/dsvm-tempest-kvm/782cf10 : SUCCESS in 1h 31m 02s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/32/check/dsvm-tempest-exe/e0f3873 : FAILURE in 1h 34m 30s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":32},{"id":"e9ae3a404acf08cee898e6b063bc2b90785b0ece","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-16 19:21:23.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/32/check/openstack-tox-lower-constraints/6313184/ : SUCCESS in 11m 10s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/32/check/openstack-tox-pep8/057d442/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/32/check/openstack-tox-py27/245f8c2/ : SUCCESS in 7m 40s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/32/check/openstack-tox-py35/d2b8718/ : SUCCESS in 8m 01s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/32/check/openstack-tox-py36/1ccf822/ : SUCCESS in 7m 46s\n- openstack-tox-docs http://logs.openstack.org/65/621465/32/check/openstack-tox-docs/8c60dce/html/ : SUCCESS in 10m 26s\n- requirements-check http://logs.openstack.org/65/621465/32/check/requirements-check/df9d91f/ : SUCCESS in 4m 14s\n- tempest-full http://logs.openstack.org/65/621465/32/check/tempest-full/9dae5eb/ : FAILURE in 1h 27m 56s\n- neutron-grenade http://logs.openstack.org/65/621465/32/check/neutron-grenade/cafde99/ : SUCCESS in 54m 24s\n- grenade-py3 http://logs.openstack.org/65/621465/32/check/grenade-py3/943bf28/ : FAILURE in 1h 08m 06s\n- tempest-full-py3 http://logs.openstack.org/65/621465/32/check/tempest-full-py3/f0694da/ : SUCCESS in 1h 31m 32s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/32/check/build-openstack-releasenotes/e8eb3d2/html/ : SUCCESS in 14m 32s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/32/check/cinder-tempest-dsvm-lvm-lio-barbican/6e7d0b8/ : SUCCESS in 1h 45m 06s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/32/check/cinder-tox-bandit-baseline/ac0ddf0/ : SUCCESS in 7m 54s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/32/check/nova-multiattach/2165401/ : SUCCESS in 1h 00m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/32/check/openstack-tox-functional-py35/dced545/ : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/32/check/openstack-tox-functional-py36/caa2c75/ : SUCCESS in 6m 39s\n- cinder-rally-task http://logs.openstack.org/65/621465/32/check/cinder-rally-task/19a169f/ : SUCCESS in 38m 34s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/32/check/openstack-tox-pylint/9dbc023/ : SUCCESS in 5m 35s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/32/check/devstack-plugin-ceph-tempest/2091a6f/ : TIMED_OUT in 2h 04m 57s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/32/check/openstack-tox-functional/dd66763/ : SUCCESS in 5m 55s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/32/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/c40020a/ : SUCCESS in 1h 03m 06s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/32/check/legacy-tempest-dsvm-lvm-multibackend/8e49d36/ : SUCCESS in 1h 14m 53s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/32/check/legacy-tempest-dsvm-full-drbd-devstack/5c688be/ : SUCCESS in 1h 20m 11s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/32/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/4c2759b/ : TIMED_OUT in 3h 03m 08s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/32/check/tempest-slow/4bbec59/ : SUCCESS in 1h 48m 55s","accounts_in_message":[],"_revision_number":32},{"id":"1070841e126b2386613c91a4f35ac9c5e5b96099","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-16 21:08:22.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/32/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/46a709c : SUCCESS in 49m 10s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/32/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/2842c95 : SUCCESS in 1h 12m 40s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":32},{"id":"152d006ed8e5be1e23fb11e36261424f80ddfde2","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-17 01:17:44.000000000","message":"Patch Set 32:\n\nrecheck","accounts_in_message":[],"_revision_number":32},{"id":"3945b0d14f0fc85edfcf20556a1acddc29d4aadc","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-17 02:05:53.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"7270edba5ce992408a93d6d0d3e45e217afe03ba","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-17 02:06:44.000000000","message":"Patch Set 33:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/33/check/dsvm-veritas-access-cinder-driver/983b99e : FAILURE in 12s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/33/check/dsvm-veritas-access-cinder-iscsi-driver/152ae10 : FAILURE in 12s","accounts_in_message":[],"_revision_number":33},{"id":"6830f9aa6575e525f287f34362a252847f61ec24","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-01-17 02:32:41.000000000","message":"Patch Set 33:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/807 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/976 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/657 : FAILURE","accounts_in_message":[],"_revision_number":33},{"id":"36b3d5cdc732beacc869ec7babd1a92a574e128d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-17 02:49:59.000000000","message":"Patch Set 33:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/33/check/fujitsu-eternus-dx-iscsi/aa3ec64 : FAILURE in 11m 00s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/33/check/fujitsu-eternus-dx-fc/5157be9 : FAILURE in 31m 24s","accounts_in_message":[],"_revision_number":33},{"id":"3c93564efd916ce32c9347f700280fc8854ed905","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-17 02:57:54.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/33/iscsi/ : SUCCESS in 51m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/33/smb/ : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":33},{"id":"48dce438afd44bec9853a7ef87663112f6714d3c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-17 03:15:44.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/33/check/dsvm-tempest-kvm/1ce63c5 : SUCCESS in 1h 06m 55s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/33/check/dsvm-tempest-exe/3b894d2 : SUCCESS in 49m 40s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":33},{"id":"1145058c239c90d15d1e6c6cf3cf63aa41aa125d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-17 03:21:23.000000000","message":"Patch Set 33:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/33/check/tempest-dsvm-full-xenial/e12bfe2/ : SUCCESS in 1h 07m 56s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/33/check/devstack-plugin-ceph-dsvm-xenial/0257218/ : SUCCESS in 1h 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"7ceaf209afde8f2c27f23ee51b9a47a9370a1f0b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-17 03:22:00.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/33/check/dsvm-tempest-cisco-zonemanager-job/e931471/ : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":33},{"id":"f711a93b15f2f767e62b3ef0693e044074967ada","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-17 03:26:04.000000000","message":"Patch Set 33:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/33 : SUCCESS","accounts_in_message":[],"_revision_number":33},{"id":"d054af033884da8f8a9735c0784a51515e0c570a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-17 03:27:20.000000000","message":"Patch Set 33:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/33/check/dsvm-tempest-infinibox-fc/3edb15e : SUCCESS in 54m 27s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/33/check/dsvm-tempest-infinibox-iscsi/2ed8d2c : FAILURE in 1h 20m 43s","accounts_in_message":[],"_revision_number":33},{"id":"747bd55ace1b27668d2a33cd861b78895743b151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 04:26:54.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/33/check/openstack-tox-lower-constraints/ee2c6fd/ : FAILURE in 11m 17s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/33/check/openstack-tox-pep8/612ccce/ : SUCCESS in 7m 27s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/33/check/openstack-tox-py27/777bdbf/ : FAILURE in 8m 58s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/33/check/openstack-tox-py35/a63e13a/ : FAILURE in 8m 47s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/33/check/openstack-tox-py36/096c5f4/ : FAILURE in 8m 40s\n- openstack-tox-docs http://logs.openstack.org/65/621465/33/check/openstack-tox-docs/1ea8b87/html/ : SUCCESS in 8m 02s\n- requirements-check http://logs.openstack.org/65/621465/33/check/requirements-check/9ade12a/ : SUCCESS in 3m 14s\n- tempest-full http://logs.openstack.org/65/621465/33/check/tempest-full/d26eaa1/ : FAILURE in 1h 08m 13s\n- neutron-grenade http://logs.openstack.org/65/621465/33/check/neutron-grenade/6c6e300/ : SUCCESS in 57m 55s\n- grenade-py3 http://logs.openstack.org/65/621465/33/check/grenade-py3/29acac6/ : SUCCESS in 1h 00m 14s\n- tempest-full-py3 http://logs.openstack.org/65/621465/33/check/tempest-full-py3/0b16b3a/ : SUCCESS in 1h 00m 50s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/33/check/build-openstack-releasenotes/5e27c09/html/ : SUCCESS in 12m 18s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/33/check/cinder-tempest-dsvm-lvm-lio-barbican/de038a8/ : SUCCESS in 2h 19m 07s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/33/check/cinder-tox-bandit-baseline/5cfd0af/ : SUCCESS in 6m 33s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/33/check/nova-multiattach/acd85e3/ : SUCCESS in 1h 00m 02s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/33/check/openstack-tox-functional-py35/5e25eb3/ : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/33/check/openstack-tox-functional-py36/445ecec/ : SUCCESS in 4m 44s\n- cinder-rally-task http://logs.openstack.org/65/621465/33/check/cinder-rally-task/551998a/ : SUCCESS in 29m 28s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/33/check/openstack-tox-pylint/e45cc38/ : SUCCESS in 4m 24s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/33/check/devstack-plugin-ceph-tempest/6456e21/ : SUCCESS in 1h 41m 39s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/33/check/openstack-tox-functional/98f9343/ : SUCCESS in 4m 52s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/33/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/7983534/ : SUCCESS in 55m 57s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/33/check/legacy-tempest-dsvm-lvm-multibackend/f452c79/ : FAILURE in 1h 17m 33s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/33/check/legacy-tempest-dsvm-full-drbd-devstack/2c5e3bd/ : FAILURE in 1h 19m 23s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/33/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/f748b4d/ : SUCCESS in 1h 34m 16s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/33/check/tempest-slow/06c5d9d/ : SUCCESS in 1h 40m 02s","accounts_in_message":[],"_revision_number":33},{"id":"2df12f59c24704126623710a3a2d11658a02f347","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-17 04:35:35.000000000","message":"Patch Set 33:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/33 : FAILURE","accounts_in_message":[],"_revision_number":33},{"id":"2131c44fe5b948c01e2f44ce66ae01ee0c8fe1da","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-17 05:05:02.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/33/check/NEC-ISCSI/1dcf1dd : SUCCESS in 1h 55m 32s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/33/check/NEC-FC/9ee857f : SUCCESS in 1h 56m 23s","accounts_in_message":[],"_revision_number":33},{"id":"63c6b224a5720f6fb52ae0f4fbbbdaaf43656a3b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-17 05:27:03.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/33/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/e324d6c : SUCCESS in 48m 52s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/33/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/6ddbb1d : SUCCESS in 1h 10m 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":33},{"id":"76064b3ecd232f1d2b62257871bbe240190c08c3","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-01-17 06:28:52.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"932fee742ed0b3776b3514ede113bca97dc3964e","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-01-17 06:37:54.000000000","message":"Patch Set 34:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/809 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/978 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/660 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"9d20b574bc19a05de507c22bd2e9174a321dac86","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-17 07:30:22.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/34/iscsi/ : SUCCESS in 1h 01m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/34/smb/ : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":34},{"id":"a794aab8ebbc7d9de840214326c98b18c10b6b88","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-17 07:31:51.000000000","message":"Patch Set 34:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/34/check/dsvm-tempest-kvm/c2c30ee : FAILURE in 41s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/34/check/dsvm-tempest-exe/26200de : SUCCESS in 1h 01m 01s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":34},{"id":"614cf8ecbf960899710f11fd4b3a6b338f00817f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-01-17 07:32:34.000000000","message":"Patch Set 34:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/34/check/fujitsu-eternus-dx-iscsi/6c530c9 : FAILURE in 55m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/34/check/fujitsu-eternus-dx-fc/b9cdbb5 : FAILURE in 54m 20s","accounts_in_message":[],"_revision_number":34},{"id":"e6a811e6a0b9e449e9ced06ade7ba43b83416f4f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-01-17 07:32:48.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/34/check-cinder/Cinder-ISER-ISCSI/fa13690 : SUCCESS in 1h 02m 51s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/34/check-cinder/Cinder-ISER-LIO/99b5276 : SUCCESS in 59m 43s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":34},{"id":"d408acd401f265fafa6b908a5b8240677058b7f0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-01-17 07:35:43.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/34/check/dsvm-tempest-infinibox-fc/1dd118d : SUCCESS in 52m 29s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/34/check/dsvm-tempest-infinibox-iscsi/5b4b5ca : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":34},{"id":"e721f9a9d740352ad9eb4dc34fcdeeaf51af5d18","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-17 07:43:43.000000000","message":"Patch Set 34:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/34/check/tempest-dsvm-full-xenial/4b61c3f/ : SUCCESS in 1h 05m 00s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/34/check/devstack-plugin-ceph-dsvm-xenial/978635c/ : SUCCESS in 1h 14m 26s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"887969192e98d4c54bb6b8a338e396259de6e9a1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-01-17 07:48:22.000000000","message":"Patch Set 34:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/34 : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"202a5035af27e14192e51843e595e4022581ff4c","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2019-01-17 07:49:07.000000000","message":"Patch Set 34: Code-Review+1","accounts_in_message":[],"_revision_number":34},{"id":"459c82602e172fe7af79cba4cad261d5fb79b7c3","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-01-17 08:07:23.000000000","message":"Patch Set 34:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/34/check/dsvm-veritas-access-cinder-driver/ee6f5f4 : SUCCESS in 1h 37m 48s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/34/check/dsvm-veritas-access-cinder-iscsi-driver/0c96cfd : FAILURE in 13s","accounts_in_message":[],"_revision_number":34},{"id":"4d046286eb2a8709f6b4e307e3629078d4e1c8e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 08:36:37.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/34/check/openstack-tox-lower-constraints/1197982/ : SUCCESS in 10m 53s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/34/check/openstack-tox-pep8/53b9594/ : SUCCESS in 7m 28s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/34/check/openstack-tox-py27/f3e59e2/ : SUCCESS in 9m 43s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/34/check/openstack-tox-py35/9ab39fc/ : SUCCESS in 8m 49s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/34/check/openstack-tox-py36/f100b97/ : SUCCESS in 7m 33s\n- openstack-tox-docs http://logs.openstack.org/65/621465/34/check/openstack-tox-docs/eb92f66/html/ : SUCCESS in 7m 50s\n- requirements-check http://logs.openstack.org/65/621465/34/check/requirements-check/66cb805/ : SUCCESS in 3m 07s\n- tempest-full http://logs.openstack.org/65/621465/34/check/tempest-full/7e15f32/ : SUCCESS in 1h 26m 18s\n- neutron-grenade http://logs.openstack.org/65/621465/34/check/neutron-grenade/dec02e8/ : SUCCESS in 1h 02m 59s\n- grenade-py3 http://logs.openstack.org/65/621465/34/check/grenade-py3/02c816f/ : SUCCESS in 49m 19s\n- tempest-full-py3 http://logs.openstack.org/65/621465/34/check/tempest-full-py3/e284168/ : SUCCESS in 1h 06m 53s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/34/check/build-openstack-releasenotes/791e2bd/html/ : SUCCESS in 12m 55s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/34/check/cinder-tempest-dsvm-lvm-lio-barbican/84292ee/ : SUCCESS in 1h 43m 52s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/34/check/cinder-tox-bandit-baseline/ac2c210/ : SUCCESS in 6m 44s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/34/check/nova-multiattach/7cf8204/ : SUCCESS in 1h 01m 38s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/34/check/openstack-tox-functional-py35/2a60cca/ : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/34/check/openstack-tox-functional-py36/64ca323/ : SUCCESS in 4m 28s\n- cinder-rally-task http://logs.openstack.org/65/621465/34/check/cinder-rally-task/0346bc5/ : SUCCESS in 37m 05s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/34/check/openstack-tox-pylint/9579850/ : SUCCESS in 4m 53s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/34/check/devstack-plugin-ceph-tempest/e9d4c45/ : TIMED_OUT in 2h 05m 04s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/34/check/openstack-tox-functional/65187c0/ : SUCCESS in 4m 10s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/34/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/ad225c8/ : SUCCESS in 1h 03m 16s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-lvm-multibackend/5382b28/ : FAILURE in 1h 34m 31s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-full-drbd-devstack/477e3a8/ : SUCCESS in 1h 31m 42s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/245c4e4/ : SUCCESS in 1h 35m 02s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/34/check/tempest-slow/33e9958/ : SUCCESS in 1h 57m 16s","accounts_in_message":[],"_revision_number":34},{"id":"dbec7b8b7d57c8becfea297180346bbda93c64bd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-17 08:48:19.000000000","message":"Patch Set 34:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/34/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9a6c18c : FAILURE in 9m 06s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/34/check/3par-fc-driver-master-client-pip-ssa02-dsvm/598f5ca : FAILURE in 4m 02s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/34/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/db34dbf : FAILURE in 2m 16s","accounts_in_message":[],"_revision_number":34},{"id":"9582e79850ecea72a5ffb8107764f9bbae3118b2","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-01-17 09:02:47.000000000","message":"Patch Set 34:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-34/nedge_iscsi : SUCCESS in 25m 14s","accounts_in_message":[],"_revision_number":34},{"id":"091b537eda904408cdbfefad142c3df85da348fc","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-17 09:04:01.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"c44f7b1466d477a8c6cc730f5b011293a2a8c154","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-01-17 09:06:13.000000000","message":"Patch Set 34:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/34/check/NEC-ISCSI/6d39c62 : FAILURE in 2h 05m 50s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/34/check/NEC-FC/ddcaafc : SUCCESS in 1h 50m 53s","accounts_in_message":[],"_revision_number":34},{"id":"8235021a824d4b434d9e16ef14d505b7aa146467","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-17 09:06:14.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/34/check/dsvm-tempest-cisco-zonemanager-job/b7c6876/ : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":34},{"id":"ffcc3675fdf8fb57ffdb859f67770b49cfb7e9cd","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-01-17 09:26:13.000000000","message":"Patch Set 34:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/34/HPMSA-FC : FAILURE in 9:51.40\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/34/HPMSA-iSCSI : FAILURE in 9:56.33","accounts_in_message":[],"_revision_number":34},{"id":"f13b97d5f33df37c98cf4be4eef09cc850bd9925","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-17 09:35:12.000000000","message":"Patch Set 34:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/793 : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"fbc013d27eec0f15d8f360ccb0fe9d51d1c9cf2d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-17 09:39:25.000000000","message":"Patch Set 34:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/34/upstream-check/cinder-cDOT-iSCSI/66eb72c/ : SUCCESS in 34m 53s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/34/upstream-check/cinder-cDOT-NFS/96168a5/ : FAILURE in 31m 38s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/34/upstream-check/cinder-cDOT-FCP/bd269b2/ : FAILURE in 50m 16s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":34},{"id":"d84bd3f8ba1d56ff5dcb4b1360f44a3040579191","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-01-17 09:51:42.000000000","message":"Patch Set 34:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-34-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"c1c906c941bd74260581cea1b3e6a1e91426971c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-01-17 10:11:33.000000000","message":"Patch Set 34:\n\nBuild succeed.\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/34/check/Dell_SC_FC/fcb0db5/Dell_SC_FC/696 : SUCCESS in 1h 06m 10s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/34/check/Dell_SC_iSCSI/c37c89b/Dell_SC_iSCSI/690 : SUCCESS in 1h 34m 16s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":34},{"id":"fa317df7b0619006123a85ef76b0e4214cb92999","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-01-17 10:19:58.000000000","message":"Patch Set 34:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/65/621465/34/check/EMC_VxFlexOS/1776a99/EMC_VxFlexOS/757 : SUCCESS in 1h 43m 02s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":34},{"id":"dc19257601119dc799d60dbcaa63bf0a95f6a14c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-01-17 10:23:07.000000000","message":"Patch Set 34:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/34/check/EMC_Unity_FC/d6155b5/EMC_Unity_FC/788 : SUCCESS in 1h 01m 03s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/34/check/EMC_Unity_iSCSI/ddec348/EMC_Unity_iSCSI/759 : FAILURE in 1h 46m 18s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":34},{"id":"543900839d824e32b307cd68863ae38e6e9ee1b8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-17 11:22:20.000000000","message":"Patch Set 34:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/34/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/2f1ef9f : FAILURE in 1h 50m 03s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/34/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/ac07816 : FAILURE in 1h 44m 45s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":34},{"id":"e718fe9f9d80e5b8eb716ab04e6c85064d9e2ebf","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-01-17 12:12:01.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/65/621465/34/upstream-eseries-check/cinder-eseries-iSCSI/2015681/ : FAILURE in 59m 16s (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/65/621465/34/upstream-eseries-check/cinder-eseries-FCP/abd0d8d/ : FAILURE in 31m 25s (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":34},{"id":"a032610e75f82b8c418dd579554243e18a6958a4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-01-17 13:02:42.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/34/upstream-solidfire-check/cinder-solidfire-iSCSI/8922a1f/ : SUCCESS in 1h 45m 39s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":34},{"id":"4f313d89fd01cdd84c365c65ca8cb42652eaacb4","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-01-17 13:36:25.000000000","message":"Patch Set 34:\n\nBuild Successful \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/797 : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"bfc03c4f2929dd1bb78be18aab7f3903f1bc1e3a","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-17 14:22:01.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"b63ed5a01edfacc9ed2784772d0410cc719520dc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-01-17 18:09:41.000000000","message":"Patch Set 34:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/34/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/772196c : SUCCESS in 2h 02m 30s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/34/check/3par-fc-driver-master-client-pip-ssa02-dsvm/ef98ed6 : FAILURE in 2h 10m 00s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/34/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/239dd8e : SUCCESS in 1h 58m 39s","accounts_in_message":[],"_revision_number":34},{"id":"69b93f4d5c80abb9e1d7f58c2b6bd73868908e22","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-01-17 18:55:17.000000000","message":"Patch Set 34: -Review-Priority\n\nRemoving priority setting.","accounts_in_message":[],"_revision_number":34},{"id":"5c70aa2aef02eadde4127516d1e89a284488bd2c","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-01-19 05:15:18.000000000","message":"Patch Set 34:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/34/Lenovo-FC : SUCCESS in 2:31:00\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/34/Lenovo-iSCSI : SUCCESS in 3:21:43","accounts_in_message":[],"_revision_number":34},{"id":"09eb14712f9667a78ee14b8cc213cbd3cea85536","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-19 12:08:24.000000000","message":"Patch Set 34:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-34 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":34},{"id":"26efabfc464173f459879a70de5e30afaa04a258","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-01-19 14:05:57.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/34/check/dsvm-tempest-storpool/b3a6ca6 : SUCCESS in 1h 41m 16s","accounts_in_message":[],"_revision_number":34},{"id":"1c15aeaf50206b25a71ba4b806f66b8aff3f6ed1","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-21 12:50:57.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"0071857c587f5c0c0b2ee67a2762db4a7be3e9be","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-21 15:49:16.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"982e51544c6ffaec5bc267520719dbbe91e1806e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-21 19:47:45.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"6966aed40df88ee93ee5209522a0010455b70eb8","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-01-25 17:26:54.000000000","message":"Patch Set 34:\n\nrecheck intel_rsd_ci","accounts_in_message":[],"_revision_number":34},{"id":"c74e16cfe2c24e8b45be18dc3b27950247a60f42","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-25 18:01:55.000000000","message":"Patch Set 34:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/34/check-not-vote/ext-cinder-zuul/e4ee13b : FAILURE in 34m 32s","accounts_in_message":[],"_revision_number":34},{"id":"1e846171d12ecf18c66fe6d3d7f879dfea6b5065","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-25 18:21:00.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/34/iscsi/ : SUCCESS in 53m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/34/smb/ : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":34},{"id":"9701e3c511aa012051071b2bb6865da01efd5fc7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-25 18:39:19.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/34/check/dsvm-tempest-cisco-zonemanager-job/c484e29/ : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":34},{"id":"8c77f02aaf2a5cf12028d0db81398a878a6e4b8f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-25 18:48:31.000000000","message":"Patch Set 34:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/34/check/tempest-dsvm-full-xenial/14ede97/ : SUCCESS in 1h 10m 13s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/34/check/devstack-plugin-ceph-dsvm-xenial/ab4d706/ : SUCCESS in 1h 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"467d7d61edc4dd951f488cbb1a8028de0bea2b18","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-01-25 19:52:44.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"6ad7e706cf43fd277efdaa967f10092f1ee9aa3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 19:54:22.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/34/check/openstack-tox-lower-constraints/faa8fc8/ : SUCCESS in 10m 38s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/34/check/openstack-tox-pep8/f9ba051/ : SUCCESS in 7m 46s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/34/check/openstack-tox-py27/b5ef7cd/ : SUCCESS in 9m 08s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/34/check/openstack-tox-py35/d20521c/ : SUCCESS in 9m 02s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/34/check/openstack-tox-py36/16459a7/ : SUCCESS in 8m 14s\n- openstack-tox-docs http://logs.openstack.org/65/621465/34/check/openstack-tox-docs/af10528/html/ : SUCCESS in 7m 56s\n- requirements-check http://logs.openstack.org/65/621465/34/check/requirements-check/1deb6d6/ : SUCCESS in 3m 14s\n- tempest-full http://logs.openstack.org/65/621465/34/check/tempest-full/e91f6d9/ : SUCCESS in 1h 46m 01s\n- neutron-grenade http://logs.openstack.org/65/621465/34/check/neutron-grenade/3fe916b/ : SUCCESS in 53m 09s\n- grenade-py3 http://logs.openstack.org/65/621465/34/check/grenade-py3/d7eafeb/ : SUCCESS in 55m 33s\n- tempest-full-py3 http://logs.openstack.org/65/621465/34/check/tempest-full-py3/a842644/ : SUCCESS in 2h 01m 34s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/34/check/build-openstack-releasenotes/cb9292b/html/ : SUCCESS in 12m 01s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/34/check/cinder-tempest-dsvm-lvm-lio-barbican/783dde3/ : FAILURE in 2h 02m 35s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/34/check/cinder-tox-bandit-baseline/0f63f33/ : SUCCESS in 8m 03s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/34/check/nova-multiattach/0a43fc1/ : SUCCESS in 1h 26m 07s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/34/check/openstack-tox-functional-py35/7ec6c08/ : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/34/check/openstack-tox-functional-py36/229e167/ : SUCCESS in 5m 35s\n- cinder-rally-task http://logs.openstack.org/65/621465/34/check/cinder-rally-task/8676870/ : SUCCESS in 32m 22s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/34/check/openstack-tox-pylint/e1220ee/ : SUCCESS in 5m 26s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/34/check/devstack-plugin-ceph-tempest/e06adf5/ : TIMED_OUT in 2h 06m 37s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/34/check/openstack-tox-functional/b198b30/ : SUCCESS in 6m 35s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/34/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/04a379a/ : SUCCESS in 1h 12m 16s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-lvm-multibackend/ea95da5/ : SUCCESS in 1h 26m 12s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-full-drbd-devstack/b0d50d1/ : FAILURE in 1h 35m 55s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/bd71142/ : FAILURE in 1h 11m 27s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/34/check/tempest-slow/caad02c/ : SUCCESS in 2h 14m 39s","accounts_in_message":[],"_revision_number":34},{"id":"5759add95bc3c2afb7f23ae8871752577afd9f2a","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-01-27 14:55:00.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12680 : FAILURE in 43m 55s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9044 : FAILURE in 1h 04m 35s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10802 : SUCCESS in 1h 05m 06s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9422 : SUCCESS in 56m 25s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10388 : FAILURE in 45m 53s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1358 : SUCCESS in 2h 23m 02s","accounts_in_message":[],"_revision_number":34},{"id":"c7dc5c1b8191bfa9c254d9a8b3ad58d184e92eee","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-01-28 15:45:14.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"84cdae4ab9802d0c132a15d1d9ca2eee0d4d15a7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-01-28 17:03:32.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/34/iscsi/ : SUCCESS in 1h 02m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/34/smb/ : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":34},{"id":"f5f2bf47c7bf367b135edbf7c983c1ec15e6b143","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-01-28 17:55:47.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/34/check/dsvm-tempest-kvm/cada226 : SUCCESS in 1h 17m 30s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/34/check/dsvm-tempest-exe/0c26c55 : FAILURE in 32m 02s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":34},{"id":"af7295ce01505d7cd5c932640dfe8808977e2a85","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-28 18:00:26.000000000","message":"Patch Set 34:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/34/check/tempest-dsvm-full-xenial/1812ac2/ : SUCCESS in 1h 27m 45s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/34/check/devstack-plugin-ceph-dsvm-xenial/f2c5dfc/ : SUCCESS in 1h 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"cee0781e2f68124fb16c635d05513a7975e86747","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-28 18:24:30.000000000","message":"Patch Set 34:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/34/check-not-vote/ext-cinder-zuul/fba479e : FAILURE in 34m 54s","accounts_in_message":[],"_revision_number":34},{"id":"57a5b9e94dc023d694221da1622794dc9cf96815","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-28 20:12:12.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/34/check/openstack-tox-lower-constraints/82c69af/ : SUCCESS in 14m 19s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/34/check/openstack-tox-pep8/858f78d/ : SUCCESS in 9m 33s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/34/check/openstack-tox-py27/0e1297d/ : SUCCESS in 10m 02s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/34/check/openstack-tox-py35/32f25f0/ : SUCCESS in 9m 55s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/34/check/openstack-tox-py36/9e94218/ : SUCCESS in 10m 28s\n- openstack-tox-docs http://logs.openstack.org/65/621465/34/check/openstack-tox-docs/7c3ddae/html/ : SUCCESS in 8m 47s\n- requirements-check http://logs.openstack.org/65/621465/34/check/requirements-check/b973e06/ : SUCCESS in 4m 12s\n- tempest-full http://logs.openstack.org/65/621465/34/check/tempest-full/9524286/ : SUCCESS in 1h 50m 44s\n- neutron-grenade http://logs.openstack.org/65/621465/34/check/neutron-grenade/ad890af/ : SUCCESS in 51m 46s\n- grenade-py3 http://logs.openstack.org/65/621465/34/check/grenade-py3/61c4c0a/ : SUCCESS in 55m 34s\n- tempest-full-py3 http://logs.openstack.org/65/621465/34/check/tempest-full-py3/1de510d/ : SUCCESS in 1h 30m 35s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/34/check/build-openstack-releasenotes/4bb5e10/html/ : SUCCESS in 13m 14s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/34/check/cinder-tempest-dsvm-lvm-lio-barbican/b281497/ : FAILURE in 2h 11m 07s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/34/check/cinder-tox-bandit-baseline/76e9710/ : SUCCESS in 30m 36s (non-voting)\n- nova-multiattach http://logs.openstack.org/65/621465/34/check/nova-multiattach/1b9f8ac/ : SUCCESS in 1h 29m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/34/check/openstack-tox-functional-py35/20d6ab5/ : SUCCESS in 30m 48s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/34/check/openstack-tox-functional-py36/c24a995/ : SUCCESS in 6m 29s\n- cinder-rally-task http://logs.openstack.org/65/621465/34/check/cinder-rally-task/7477bf3/ : SUCCESS in 43m 16s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/34/check/openstack-tox-pylint/c890912/ : SUCCESS in 6m 15s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/34/check/devstack-plugin-ceph-tempest/feea825/ : TIMED_OUT in 2h 05m 48s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/34/check/openstack-tox-functional/b2af03d/ : SUCCESS in 6m 11s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/34/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/595f669/ : SUCCESS in 56m 25s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-lvm-multibackend/9eed476/ : FAILURE in 1h 38m 06s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-full-drbd-devstack/c080cf9/ : FAILURE in 1h 37m 20s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/fe729c3/ : SUCCESS in 1h 46m 01s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/34/check/tempest-slow/306d2d4/ : SUCCESS in 2h 11m 00s","accounts_in_message":[],"_revision_number":34},{"id":"a11b660a41b3a5ebcc3c7f97df9cfa77866833ef","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-01-28 21:02:58.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/34/check/dsvm-tempest-cisco-zonemanager-job/3862111/ : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":34},{"id":"af54d8ffa97ff427000df74e46969c1bb8f362f9","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-01-29 13:14:58.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/12709 : FAILURE in 36m 06s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9071 : SUCCESS in 57m 35s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/10829 : SUCCESS in 1h 01m 40s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9449 : SUCCESS in 57m 42s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10416 : FAILURE in 37m 14s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1381 : SUCCESS in 1h 58m 56s","accounts_in_message":[],"_revision_number":34},{"id":"fa6b5ce7e2f53b47058b42f7a00eaa6e32849479","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-02-21 03:18:07.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"34e2c1dace6aae69f8f47807c66bce0054696b37","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-02-21 03:20:10.000000000","message":"Patch Set 34:\n\nrecheck intel_rsd_ci","accounts_in_message":[],"_revision_number":34},{"id":"cd3336f249862be9608fe4be2a0b2b0ef103aa65","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-21 03:56:56.000000000","message":"Patch Set 34:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/34/check-not-vote/ext-cinder-zuul/1e8e4b7 : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":34},{"id":"894d4000f46e7f73c0b677fbfa416bcba786e44a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-02-21 04:15:13.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/34/check/dsvm-tempest-kvm/da19c91 : SUCCESS in 56m 32s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/34/check/dsvm-tempest-exe/81dc7f0 : FAILURE in 11m 05s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":34},{"id":"2f91e7431038fee901bc9de113bd6b04303b17ca","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-02-21 04:19:06.000000000","message":"Patch Set 34:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/34/iscsi/ : FAILURE in 45m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/34/smb/ : FAILURE in 1h 00m 20s","accounts_in_message":[],"_revision_number":34},{"id":"84edfa751b146b717648a8b1ae2b6c5b13ef9dc7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-21 04:56:11.000000000","message":"Patch Set 34:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/34/check/tempest-dsvm-full-xenial/545dad4/ : UNSTABLE in 1h 09m 25s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/34/check/devstack-plugin-ceph-dsvm-xenial/cb973bd/ : FAILURE in 1h 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"b1a425f36140947b7e05daa6b8ccfe35f53047b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 05:36:36.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/34/check/openstack-tox-lower-constraints/21b6fa0/ : SUCCESS in 11m 10s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/34/check/openstack-tox-pep8/8ca9f4f/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/34/check/openstack-tox-py27/adb1d8a/ : SUCCESS in 9m 40s\n- openstack-tox-py35 http://logs.openstack.org/65/621465/34/check/openstack-tox-py35/0251a08/ : SUCCESS in 10m 36s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/34/check/openstack-tox-py36/38530b9/ : SUCCESS in 8m 53s\n- openstack-tox-docs http://logs.openstack.org/65/621465/34/check/openstack-tox-docs/172aa0c/html/ : SUCCESS in 8m 43s\n- requirements-check http://logs.openstack.org/65/621465/34/check/requirements-check/2a69876/ : SUCCESS in 4m 36s\n- tempest-full http://logs.openstack.org/65/621465/34/check/tempest-full/034d1fa/ : SUCCESS in 1h 50m 12s\n- neutron-grenade http://logs.openstack.org/65/621465/34/check/neutron-grenade/2902cbe/ : SUCCESS in 55m 18s\n- grenade-py3 http://logs.openstack.org/65/621465/34/check/grenade-py3/7733289/ : SUCCESS in 58m 41s\n- tempest-full-py3 http://logs.openstack.org/65/621465/34/check/tempest-full-py3/5234727/ : SUCCESS in 1h 36m 38s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/34/check/build-openstack-releasenotes/0a643ec/html/ : SUCCESS in 11m 05s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/34/check/cinder-tempest-dsvm-lvm-lio-barbican/75dba1a/ : SUCCESS in 1h 59m 11s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/34/check/cinder-tox-bandit-baseline/66f0129/ : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py35 http://logs.openstack.org/65/621465/34/check/openstack-tox-functional-py35/effcf1a/ : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/34/check/openstack-tox-functional-py36/87e0eac/ : SUCCESS in 6m 30s\n- cinder-rally-task http://logs.openstack.org/65/621465/34/check/cinder-rally-task/29e140a/ : SUCCESS in 35m 57s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/34/check/openstack-tox-pylint/5112baa/ : SUCCESS in 4m 23s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/34/check/devstack-plugin-ceph-tempest/8cd778b/ : TIMED_OUT in 2h 04m 39s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/34/check/openstack-tox-functional/4b5db0f/ : SUCCESS in 4m 49s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/34/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/e44de8d/ : SUCCESS in 1h 05m 13s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-lvm-multibackend/4c431ef/ : SUCCESS in 1h 31m 11s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-full-drbd-devstack/8ac6fdf/ : SUCCESS in 1h 28m 31s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/34/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/b8d824f/ : SUCCESS in 1h 41m 32s (non-voting)\n- tempest-slow http://logs.openstack.org/65/621465/34/check/tempest-slow/d250d1c/ : SUCCESS in 2h 12m 20s","accounts_in_message":[],"_revision_number":34},{"id":"5bdfffdaf01d4258416e88d6316c0d1baf498811","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-02-21 05:45:18.000000000","message":"Patch Set 34:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/34/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/7f7101b : FAILURE in 1h 42m 46s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/34/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/9fe5ec4 : FAILURE in 1h 49m 26s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":34},{"id":"287212e226e776dbfb31bec2692524df591029b7","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-02-21 06:28:16.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://10.17.200.215/65/621465/34/check/synology-dsm-cinder-iscsi/446066e : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":34},{"id":"a401fcdd40c783a2b0cc0d9c5afd6edf9fe8406b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-02-21 06:29:45.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/34/upstream-solidfire-check/cinder-solidfire-iSCSI/b63f03c/ : SUCCESS in 42m 51s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":34},{"id":"8268c9cff1c51969337222786f5a20936a4639f8","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-02-21 06:56:08.000000000","message":"Patch Set 34:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-34-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"1505ca6366999c7923eaf0ad8a992230c045680b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-02-21 07:11:21.000000000","message":"Patch Set 34:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/34/Lenovo-FC : FAILURE in 21:54.26\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/34/Lenovo-iSCSI : FAILURE in 22:14.43","accounts_in_message":[],"_revision_number":34},{"id":"85507ad64c6da674066df6f688bb3f46397a216d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-21 07:23:39.000000000","message":"Patch Set 34:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/65/621465/34/check/EMC_VxFlexOS/ab7f1a2/EMC_VxFlexOS/931 : SUCCESS in 1h 46m 47s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":34},{"id":"8280c5d293cfeea7b1d9d51558bd84421987f71f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-02-21 07:40:06.000000000","message":"Patch Set 34:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/34/HPMSA-FC : SUCCESS in 42:16.96\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/34/HPMSA-iSCSI : SUCCESS in 41:36.31","accounts_in_message":[],"_revision_number":34},{"id":"20b8a198e93ba0545c832351036a7b9668d402d0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-02-21 07:50:37.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/34/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ef483ae : SUCCESS in 2h 11m 28s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/34/check/3par-fc-driver-master-client-pip-ssa02-dsvm/82f9e86 : SUCCESS in 2h 07m 31s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/34/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/4851348 : SUCCESS in 2h 01m 53s","accounts_in_message":[],"_revision_number":34},{"id":"e5f4d024d558ecad96fa2687ce5439b6e0fd9d24","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-02-21 08:58:25.000000000","message":"Patch Set 34:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/34/upstream-check/cinder-cDOT-iSCSI/50b9576/ : POST_FAILURE in 17m 17s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/34/upstream-check/cinder-cDOT-NFS/0dca6fe/ : SUCCESS in 55m 06s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/34/upstream-check/cinder-cDOT-FCP/5c5b502/ : FAILURE in 50m 06s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":34},{"id":"fad6fd637a8cd5316829844e156243d1420bbdaa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-02-21 09:27:39.000000000","message":"Patch Set 34:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/34/check/dsvm-tempest-cisco-zonemanager-job/eaf574c/ : FAILURE in 53m 42s","accounts_in_message":[],"_revision_number":34},{"id":"7f2054b901872e79bcc4c14e3462657326943b7a","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-02-21 11:21:27.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/34/check/kaminario-dsvm-tempest-full-iscsi/6cc2762 : SUCCESS in 2h 08m 38s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/34/check/kaminario-dsvm-tempest-full-FC/46e5290 : SUCCESS in 1h 49m 38s","accounts_in_message":[],"_revision_number":34},{"id":"70143629a4aa0583ab3ffadcc8353848ce393fb9","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-02-21 11:22:00.000000000","message":"Patch Set 34:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/1307 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"7906de8e9b165c7eb351cab57c0c96adbe53b857","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-02-21 12:57:27.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"e240e7e41a8b26a4f255b55d98f85bff43d3de1b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-02-21 20:11:56.000000000","message":"Patch Set 34:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/34/check/Dell_SC_FC/6447cd1/Dell_SC_FC/892 : FAILURE in 3h 08m 52s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/34/check/Dell_SC_iSCSI/30034db/Dell_SC_iSCSI/889 : FAILURE in 1h 41m 19s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":34},{"id":"2612f9f08a4bda19a288fc2d84582b9995fb210b","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-02-22 02:39:31.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/13013 : FAILURE in 36m 59s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9381 : FAILURE in 31s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/11105 : SUCCESS in 1h 08m 49s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9756 : FAILURE in 2h 02m 53s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10728 : FAILURE in 36m 16s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1621 : SUCCESS in 1h 59m 52s","accounts_in_message":[],"_revision_number":34},{"id":"24998e18f95ae4bb1bb7e1dc7dc10b9e5e6aa8a6","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-02-23 13:29:01.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"6ceed8c225ab9c21eeadf80f1c55f0a3918b74a0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-02-23 19:26:16.000000000","message":"Patch Set 34:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/34/check/EMC_XtremIO_FC/faa9505/EMC_XtremIO_FC/1683 : SUCCESS in 1h 30m 01s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/34/check/EMC_XtremIO_iSCSI/8afe003/EMC_XtremIO_iSCSI/1686 : SUCCESS in 1h 27m 04s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":34},{"id":"7d2bbd306aa3e98fb4eb9d4e50df2e73e9ea0812","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-02-25 19:17:53.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"bd0c440ef883a69faf452753563ac43254dc7e8a","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-02-26 14:21:39.000000000","message":"Patch Set 34: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":34},{"id":"8b21245a8290bf249edb7815faee0f36225af106","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-02-26 16:20:58.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"b127b5434276c2ca222c3ff1e6b1c5c3d42e4c52","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-02-27 17:22:21.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"9f13399ecea0abbe4f1716d05b94872627dc1cc2","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-02-27 21:59:04.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"8bb8878cf9dae516ebc88d15a276b40db5612e9f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-02-28 00:17:10.000000000","message":"Patch Set 34:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"7329af30720516e9d190f823696290dc166b21ea","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-02-28 21:02:15.000000000","message":"Patch Set 34:\n\nBuild succeed.\n\n- EMC_VMAX_FC http://publiclogs.emc.com/621465/34/EMC_VMAX_FC/924 : SUCCESS in 7h 26m 48s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/621465/34/EMC_VMAX_iSCSI/947 : SUCCESS in 7h 28m 15s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":34},{"id":"01e2bd0235faa1aea72ee11205db88994e6ac7d4","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-03-01 05:40:28.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"e5f5201e1d1e9d9b42f01b1342901f0946220ba2","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-03-01 05:44:05.000000000","message":"Patch Set 35:\n\n(4 comments)\n\nThank you for the comments, I addressed the specified issues.","accounts_in_message":[],"_revision_number":35},{"id":"34166bd36db5f04824cefd7611b52422eccb09d9","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-03-01 05:53:24.000000000","message":"Patch Set 35:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/991 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1276 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1115 : SUCCESS","accounts_in_message":[],"_revision_number":35},{"id":"beefbb22ab4e2f5708614994e11f0d4e7d01b40b","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-03-01 06:34:24.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/35/check/dsvm-tempest-kvm/378991d : SUCCESS in 53m 10s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/35/check/dsvm-tempest-exe/e68cf43 : FAILURE in 1m 16s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"f1041f8b88c09f1d75389a419882e4b43e36ab67","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-01 06:41:26.000000000","message":"Patch Set 35:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/35/check-not-vote/ext-cinder-zuul/01e9620 : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":35},{"id":"67062c8d6e8d1a0fe224084c12f424dd24940e97","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-03-01 06:53:37.000000000","message":"Patch Set 35:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/35/iscsi/ : SUCCESS in 56m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/35/smb/ : FAILURE in 50m 58s","accounts_in_message":[],"_revision_number":35},{"id":"33ac681d99e2f2f72c9f23c5ef827d2f89f599b7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-03-01 06:56:53.000000000","message":"Patch Set 35:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/35/check/fujitsu-eternus-dx-iscsi/6902886 : FAILURE in 1h 13m 36s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/35/check/fujitsu-eternus-dx-fc/4ca6b28 : FAILURE in 58m 17s","accounts_in_message":[],"_revision_number":35},{"id":"14cf63151a6e8ebeb52107b7141ad2ed47c1319d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-03-01 07:20:18.000000000","message":"Patch Set 35:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/35 : SUCCESS","accounts_in_message":[],"_revision_number":35},{"id":"a130f48509b450d9275f97a10c80cfd9d64d99d7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-03-01 07:25:54.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/35/check/dsvm-tempest-storpool/7187173 : SUCCESS in 1h 44m 59s","accounts_in_message":[],"_revision_number":35},{"id":"028535dfb345d07cca34b3bc434c4a1059c82154","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-03-01 07:30:21.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/35/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/a4c4d9a : FAILURE in 1h 49m 30s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/35/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e449365 : SUCCESS in 1h 01m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":35},{"id":"a86cf6b4f91af464ffa6d2feac447a70ebd805f2","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-03-01 07:48:15.000000000","message":"Patch Set 35:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/35 : FAILURE","accounts_in_message":[],"_revision_number":35},{"id":"24c4122257d4aac147b0e23e4f5b995773b31966","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-03-01 08:05:32.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/35/check/dsvm-tempest-cisco-zonemanager-job/401fa1f/ : FAILURE in 55m 41s","accounts_in_message":[],"_revision_number":35},{"id":"4cd1c863c605a75ae06061c21c067a31601d8773","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-01 08:52:17.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/35/check/openstack-tox-lower-constraints/ded0d69/ : SUCCESS in 10m 27s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/35/check/openstack-tox-pep8/7f2246f/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/35/check/openstack-tox-py27/9fbd1ce/ : SUCCESS in 10m 25s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/35/check/openstack-tox-py36/4255fa4/ : SUCCESS in 8m 13s\n- openstack-tox-docs http://logs.openstack.org/65/621465/35/check/openstack-tox-docs/03b848c/ : POST_FAILURE in 4m 15s\n- requirements-check http://logs.openstack.org/65/621465/35/check/requirements-check/99244b9/ : SUCCESS in 2m 55s\n- tempest-full http://logs.openstack.org/65/621465/35/check/tempest-full/2013366/ : SUCCESS in 1h 37m 19s\n- neutron-grenade http://logs.openstack.org/65/621465/35/check/neutron-grenade/0716dcb/ : SUCCESS in 55m 08s\n- grenade-py3 http://logs.openstack.org/65/621465/35/check/grenade-py3/7b7c8e8/ : SUCCESS in 1h 00m 22s\n- tempest-full-py3 http://logs.openstack.org/65/621465/35/check/tempest-full-py3/a9ed62b/ : SUCCESS in 1h 23m 49s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/35/check/build-openstack-releasenotes/1c4402f/html/ : SUCCESS in 11m 19s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/35/check/cinder-tempest-dsvm-lvm-lio-barbican/ae91855/ : SUCCESS in 1h 37m 14s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/35/check/cinder-tox-bandit-baseline/e0e731d/ : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/35/check/openstack-tox-functional-py36/1b66e80/ : SUCCESS in 4m 50s\n- cinder-rally-task http://logs.openstack.org/65/621465/35/check/cinder-rally-task/6d29ffd/ : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/35/check/openstack-tox-pylint/d02409f/ : SUCCESS in 4m 23s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/35/check/devstack-plugin-ceph-tempest/47cf768/ : SUCCESS in 1h 24m 21s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/35/check/openstack-tox-functional/22249ec/ : SUCCESS in 4m 51s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/35/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/5625385/ : SUCCESS in 1h 04m 01s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/35/check/legacy-tempest-dsvm-lvm-multibackend/0342880/ : SUCCESS in 1h 30m 07s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/35/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/3663ec2/ : TIMED_OUT in 3h 02m 31s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/35/check/tempest-slow-py3/a4be10f/ : SUCCESS in 2h 05m 45s","accounts_in_message":[],"_revision_number":35},{"id":"98e44b307db2780b3bb8c8d158648447b3b0b2ac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-03-01 11:12:47.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_FC http://publiclogs.emc.com/65/621465/35/check/EMC_Unity_FC/488e4fd/EMC_Unity_FC/1262 : FAILURE in 1h 12m 13s\n- EMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/35/check/EMC_Unity_iSCSI/6576941/EMC_Unity_iSCSI/1199 : FAILURE in 1h 48m 20s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":35},{"id":"e5cc59a5fd34b5a9b9bcf3ad07de4f9123a88d0b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-03-01 15:45:31.000000000","message":"Patch Set 35:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/35/check/EMC_XtremIO_FC/36747a2/EMC_XtremIO_FC/1783 : SUCCESS in 1h 30m 13s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/35/check/EMC_XtremIO_iSCSI/45777bb/EMC_XtremIO_iSCSI/1762 : SUCCESS in 1h 25m 58s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":35},{"id":"19e11a6bfc315316630f1f73ce9a390530f8cf2f","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-03-01 17:06:00.000000000","message":"Patch Set 35:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/35/check/dsvm-veritas-access-cinder-driver/5a19ecb : FAILURE in 6m 04s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/35/check/dsvm-veritas-access-cinder-iscsi-driver/c09313f : FAILURE in 6m 09s","accounts_in_message":[],"_revision_number":35},{"id":"e078f562ba22fbd242beb8528f89ffd7a130382b","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-03-01 19:00:34.000000000","message":"Patch Set 35:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/35/check/Cinder-AS13000/543acc5 : SUCCESS in 2h 19m 45s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/35/check/cinder-test-fc/fb9cb8c : SUCCESS in 1h 03m 04s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/35/check/cinder-test-iscsi/76307f6 : FAILURE in 37m 36s","accounts_in_message":[],"_revision_number":35},{"id":"b06cbdb08922bce69c01c0e15ebf559c0a203148","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-03-02 08:06:20.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/35/check/NEC-ISCSI/1ece7d4 : SUCCESS in 2h 46m 47s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/35/check/NEC-FC/79a9ca0 : SUCCESS in 2h 06m 28s","accounts_in_message":[],"_revision_number":35},{"id":"f1cf7e608a1cec6b777db00b9e46000c4b55af21","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-03-06 01:46:49.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"4ab1baf82e9e5030b788bc0abc5b555302463e7e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-06 02:23:33.000000000","message":"Patch Set 35:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/35/check-not-vote/ext-cinder-zuul/19edcd5 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":35},{"id":"02df864dce724bb183f673cbeed1bb505e675c46","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-03-06 02:33:28.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/35/check/dsvm-tempest-kvm/26c1293 : FAILURE in 43m 17s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/35/check/dsvm-tempest-exe/8cae24c : FAILURE in 36m 55s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"48e50181de82b2c64c5cc7db2e32e1c4a14a0c2c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-03-06 02:42:55.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/35/iscsi/ : SUCCESS in 55m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/35/smb/ : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":35},{"id":"172848bbdc783adfaf9f2395f505ba456fe636e1","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-03-06 03:36:43.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/35/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/55609fb : FAILURE in 1h 49m 32s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/35/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a83c086 : SUCCESS in 1h 08m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":35},{"id":"66a22f6f487b2a0e433233dfa7006127ddb04bc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 04:40:13.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/35/check/openstack-tox-lower-constraints/e9c44d7/ : SUCCESS in 11m 25s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/35/check/openstack-tox-pep8/3fd655e/ : SUCCESS in 7m 35s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/35/check/openstack-tox-py27/b62353d/ : SUCCESS in 9m 34s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/35/check/openstack-tox-py36/f7e5c02/ : SUCCESS in 8m 17s\n- openstack-tox-docs http://logs.openstack.org/65/621465/35/check/openstack-tox-docs/3d24948/ : POST_FAILURE in 4m 35s\n- requirements-check http://logs.openstack.org/65/621465/35/check/requirements-check/3811f8e/ : SUCCESS in 3m 05s\n- tempest-full http://logs.openstack.org/65/621465/35/check/tempest-full/f2308df/ : SUCCESS in 1h 43m 01s\n- neutron-grenade http://logs.openstack.org/65/621465/35/check/neutron-grenade/eac674a/ : SUCCESS in 50m 36s\n- grenade-py3 http://logs.openstack.org/65/621465/35/check/grenade-py3/9084a39/ : SUCCESS in 54m 10s\n- tempest-full-py3 http://logs.openstack.org/65/621465/35/check/tempest-full-py3/2887dd2/ : FAILURE in 1h 04m 04s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/35/check/build-openstack-releasenotes/7fd9311/html/ : SUCCESS in 12m 27s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/35/check/cinder-tempest-dsvm-lvm-lio-barbican/5405bb6/ : SUCCESS in 1h 46m 10s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/35/check/cinder-tox-bandit-baseline/349a1ee/ : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/35/check/openstack-tox-functional-py36/bd8dba5/ : SUCCESS in 6m 27s\n- cinder-rally-task http://logs.openstack.org/65/621465/35/check/cinder-rally-task/01e3511/ : SUCCESS in 46m 58s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/35/check/openstack-tox-pylint/1f453c9/ : SUCCESS in 4m 39s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/35/check/devstack-plugin-ceph-tempest/f074cfc/ : SUCCESS in 1h 27m 41s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/35/check/openstack-tox-functional/0da28c9/ : SUCCESS in 5m 43s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/35/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/f94fe42/ : SUCCESS in 59m 46s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/35/check/legacy-tempest-dsvm-lvm-multibackend/f296315/ : SUCCESS in 1h 33m 12s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/35/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/bc44a8e/ : SUCCESS in 1h 36m 26s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/35/check/tempest-slow-py3/18edac7/ : SUCCESS in 2h 09m 49s","accounts_in_message":[],"_revision_number":35},{"id":"ef5a27594bcea84dc6cc5e87037f5a0930039235","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-03-06 05:22:16.000000000","message":"Patch Set 35:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/35/check/Cinder-AS13000/87a889d : SUCCESS in 2h 11m 26s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/35/check/cinder-test-fc/1a01748 : SUCCESS in 58m 09s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/35/check/cinder-test-iscsi/dc186d9 : FAILURE in 14m 49s","accounts_in_message":[],"_revision_number":35},{"id":"65cfd4999d5aaa83b594dd7f95a3daa3151b3ff6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-03-06 09:28:45.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/35/check/dsvm-tempest-cisco-zonemanager-job/b47420e/ : FAILURE in 56m 26s","accounts_in_message":[],"_revision_number":35},{"id":"b87aaad7f98f8b7ad5c5bc488f048202f24f4d99","author":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"date":"2019-03-06 18:32:26.000000000","message":"Patch Set 35:\n\nBuild failed. To recheck use \u0027datacore: recheck\u0027. For information about DataCore CI see https://wiki.openstack.org/wiki/ThirdPartySystems/DataCore_CI\n\n- dsvm-tempest-datacore-iscsi-driver http://173.221.96.158/65/621465/35/check/dsvm-tempest-datacore-iscsi-driver/962f4fb : FAILURE in 45m 37s\n- dsvm-tempest-datacore-fc-driver http://173.221.96.158/65/621465/35/check/dsvm-tempest-datacore-fc-driver/5fbb379 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":35},{"id":"fc068c10cc204d448aecc14e629a1d7a63c5be92","author":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"date":"2019-03-08 02:18:26.000000000","message":"Patch Set 35: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"ac889e8c0bcba7286df3f1a83c36e6ad88e75a0d","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-03-08 23:03:59.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"fa3a1e7e226ed1bffa5b5e0ce031901c6304344b","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-03-08 23:21:52.000000000","message":"Patch Set 36:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1049 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1321 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1160 : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"f46f1719d0e63bb253c37677fb47be8db7843fcb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-08 23:43:53.000000000","message":"Patch Set 36:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/36/check-not-vote/ext-cinder-zuul/4467319 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":36},{"id":"74d59a979d4fceb7fe70b8de46fedf3ed64a7005","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 23:50:56.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/36/check/dsvm-tempest-kvm/9d3e8e0 : FAILURE in 1m 10s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/36/check/dsvm-tempest-exe/60942c0 : FAILURE in 37m 48s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":36},{"id":"cc0e2f7a61374f90e10d899aae701e3da0697cc3","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-03-08 23:53:56.000000000","message":"Patch Set 36:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-03-08_15-04-05/ : FAILURE","accounts_in_message":[],"_revision_number":36},{"id":"89562138171aeebb92c108ed9f34662a233eca7a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-03-08 23:58:07.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/36/iscsi/ : SUCCESS in 53m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/36/smb/ : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":36},{"id":"c839670e0c3ebf4a81dbbe49e961358c9bbdb267","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-03-09 00:09:31.000000000","message":"Patch Set 36:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-cinder-driver http://52.42.67.99/65/621465/36/check/dsvm-veritas-access-cinder-driver/f4f9c8a : FAILURE in 1h 04m 42s\n- dsvm-veritas-access-cinder-iscsi-driver http://52.42.67.99/65/621465/36/check/dsvm-veritas-access-cinder-iscsi-driver/51a05aa : FAILURE in 55m 29s","accounts_in_message":[],"_revision_number":36},{"id":"b0a65ac758d6113388f9b4e8f04f56e498f1d50a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-09 00:18:21.000000000","message":"Patch Set 36:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/36/check/tempest-dsvm-full-xenial/08655a2/ : SUCCESS in 1h 06m 50s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/36/check/devstack-plugin-ceph-dsvm-xenial/b08cb29/ : SUCCESS in 1h 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"44dd4a6c14fbd3a40241222cf1645a8815cfbd37","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-03-09 00:21:59.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/36/check/dsvm-tempest-infinibox-fc/fca3848 : SUCCESS in 52m 20s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/36/check/dsvm-tempest-infinibox-iscsi/db0b2d3 : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":36},{"id":"c45bd5ce453968143a0ee72cf37f059ac99b5210","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-03-09 00:31:52.000000000","message":"Patch Set 36:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/36 : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"2139e066d52af0c5a90d1fb1ef42cfd49f220861","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-03-09 00:42:58.000000000","message":"Patch Set 36:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/36/check/fujitsu-eternus-dx-iscsi/fd3d79f : SUCCESS in 1h 34m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/36/check/fujitsu-eternus-dx-fc/59691f1 : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":36},{"id":"1073661dd33146b6fd3788f5f8ecb329d2a090c4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-03-09 00:49:49.000000000","message":"Patch Set 36:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/36/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/fb30396 : FAILURE in 1h 45m 25s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/36/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/8a021ba : SUCCESS in 1h 13m 02s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":36},{"id":"b53b1c7714ce27219547170088bd09574fa9dc22","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-03-09 01:00:11.000000000","message":"Patch Set 36:\n\n(2 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"cb9a077e73c1cc4214b8f4d5a794ae9add6085ff","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-03-09 01:20:18.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/36/check/Cinder-AS13000/1bf765d : SUCCESS in 2h 13m 34s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/36/check/cinder-test-fc/96506a5 : SUCCESS in 58m 42s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/36/check/cinder-test-iscsi/8e3aa93 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":36},{"id":"17af8ca6b52bfa22f90b834ce22fe62405d1b35a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-09 01:30:35.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/36/check/openstack-tox-lower-constraints/ac471ee/ : SUCCESS in 12m 26s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/36/check/openstack-tox-pep8/a2f6a0e/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/36/check/openstack-tox-py27/b6345a8/ : SUCCESS in 9m 07s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/36/check/openstack-tox-py36/b016ec8/ : SUCCESS in 9m 36s\n- openstack-tox-docs http://logs.openstack.org/65/621465/36/check/openstack-tox-docs/e648d9d/html/ : SUCCESS in 8m 37s\n- requirements-check http://logs.openstack.org/65/621465/36/check/requirements-check/f67f4dc/ : SUCCESS in 3m 01s\n- tempest-full http://logs.openstack.org/65/621465/36/check/tempest-full/3c8d14b/ : SUCCESS in 1h 46m 58s\n- neutron-grenade http://logs.openstack.org/65/621465/36/check/neutron-grenade/beccbd3/ : SUCCESS in 57m 04s\n- grenade-py3 http://logs.openstack.org/65/621465/36/check/grenade-py3/4c0fe6d/ : SUCCESS in 1h 05m 56s\n- tempest-full-py3 http://logs.openstack.org/65/621465/36/check/tempest-full-py3/641e1f6/ : SUCCESS in 1h 25m 45s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/36/check/build-openstack-releasenotes/63bc299/html/ : SUCCESS in 15m 08s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/36/check/cinder-tempest-dsvm-lvm-lio-barbican/07dcca5/ : SUCCESS in 1h 44m 54s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/36/check/cinder-tox-bandit-baseline/d0d40df/ : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/36/check/openstack-tox-functional-py36/2632d70/ : SUCCESS in 4m 58s\n- cinder-rally-task http://logs.openstack.org/65/621465/36/check/cinder-rally-task/c2f2576/ : SUCCESS in 37m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/36/check/openstack-tox-pylint/41b0c86/ : SUCCESS in 6m 27s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/36/check/devstack-plugin-ceph-tempest/322bd93/ : SUCCESS in 1h 37m 24s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/36/check/openstack-tox-functional/812ac2a/ : SUCCESS in 5m 49s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/36/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/c498341/ : SUCCESS in 1h 10m 01s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/36/check/legacy-tempest-dsvm-lvm-multibackend/26309e6/ : SUCCESS in 1h 45m 17s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/36/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/5f6d316/ : SUCCESS in 1h 46m 18s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/36/check/tempest-slow-py3/db44f21/ : SUCCESS in 2h 21m 14s","accounts_in_message":[],"_revision_number":36},{"id":"2822a9773022068ca6bf06047d1a7b1113466380","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-03-09 02:16:57.000000000","message":"Patch Set 36:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-36/ns5_iscsi : SUCCESS in 40m 14s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-36/ns5_nfs : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":36},{"id":"7d942c9aa515da4d7d62ed565120d41b2354c7dd","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-03-09 02:25:02.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/65/621465/36/check/synology-dsm-cinder-iscsi/593f5c7 : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":36},{"id":"198150c41a12478f2398e493c0abd3f2e0f42722","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-09 02:34:50.000000000","message":"Patch Set 36:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-36 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":36},{"id":"ac3b8f655a15bf5fc4f37f904f1aeb78cc1f224a","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-03-09 02:48:14.000000000","message":"Patch Set 36:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-36 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI for rechecking and info.","accounts_in_message":[],"_revision_number":36},{"id":"2ecc811723d30918a85dc18c7d32a695bda414a4","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-03-09 02:50:57.000000000","message":"Patch Set 36:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-36-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"2991a31f8789850eb9135e65a153ccf8db425c63","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-03-09 02:57:56.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/36/upstream-solidfire-check/cinder-solidfire-iSCSI/f8d8929/ : FAILURE in 1h 25m 59s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":36},{"id":"45d39c0dbddeca2db0ab04fe973d6605850475ba","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-03-09 03:12:46.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/36/upstream-check/cinder-cDOT-iSCSI/2c53148/ : SUCCESS in 1h 41m 16s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/36/upstream-check/cinder-cDOT-NFS/7b2d97b/ : SUCCESS in 40m 50s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/36/upstream-check/cinder-cDOT-FCP/8366e60/ : SUCCESS in 1h 08m 07s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":36},{"id":"71f429aa0861db2ba2ba9d5948e510092c71107e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-09 03:15:20.000000000","message":"Patch Set 36:\n\nBuild succeed.\n\n- EMC_VxFlexOS http://publiclogs.emc.com/65/621465/36/check/EMC_VxFlexOS/ca4774f/EMC_VxFlexOS/1031 : SUCCESS in 1h 44m 18s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":36},{"id":"04cbe794b1c1c88a5be9de57621f0e094657b222","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-03-09 03:20:22.000000000","message":"Patch Set 36:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/36/Lenovo-FC : SUCCESS in 39:52.00\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/36/Lenovo-iSCSI : SUCCESS in 38:08.40","accounts_in_message":[],"_revision_number":36},{"id":"ae8949e08808a06074782e15e965529793c25c78","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-03-09 03:25:09.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/36/check/kaminario-dsvm-tempest-full-iscsi/a5f0f32 : SUCCESS in 1h 54m 00s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/36/check/kaminario-dsvm-tempest-full-FC/61d11f2 : SUCCESS in 1h 48m 41s","accounts_in_message":[],"_revision_number":36},{"id":"9f3373d6f126427bf465997fa2d6df2e7361837f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-03-09 03:26:41.000000000","message":"Patch Set 36:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/36/HPMSA-FC : SUCCESS in 42:28.97\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/36/HPMSA-iSCSI : SUCCESS in 41:52.85","accounts_in_message":[],"_revision_number":36},{"id":"5ddf649506096a2276c0c4142fdcf19bb61744ca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-03-09 03:52:46.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/36/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f98b924 : SUCCESS in 2h 17m 10s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/36/check/3par-fc-driver-master-client-pip-ssa02-dsvm/9c52108 : SUCCESS in 2h 19m 31s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/36/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/e880e75 : SUCCESS in 2h 12m 39s","accounts_in_message":[],"_revision_number":36},{"id":"d0307aca96c9cf58e00d195341aa944af2cd3f7e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-03-09 04:34:22.000000000","message":"Patch Set 36:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- Dell_SC_FC http://publiclogs.emc.com/65/621465/36/check/Dell_SC_FC/35313a0/Dell_SC_FC/984 : SUCCESS in 2h 50m 27s\n- Dell_SC_iSCSI http://publiclogs.emc.com/65/621465/36/check/Dell_SC_iSCSI/8ac1914/Dell_SC_iSCSI/979 : FAILURE in 3h 03m 26s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":36},{"id":"4f1b794a982e0e6262339061285eae8e360eb71e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-03-09 07:32:11.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/36/check/dsvm-tempest-cisco-zonemanager-job/5db104a/ : FAILURE in 58m 45s","accounts_in_message":[],"_revision_number":36},{"id":"c0cb86c846dbf5d9c063dc06f4ab48408efcd03e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-03-09 09:28:26.000000000","message":"Patch Set 36:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/36/check/EMC_XtremIO_FC/5e063ed/EMC_XtremIO_FC/1892 : FAILURE in 2m 25s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/36/check/EMC_XtremIO_iSCSI/a9833b2/EMC_XtremIO_iSCSI/1871 : FAILURE in 1m 07s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":36},{"id":"1edbccc34ae0bb48785a99c9f3a4cdad20f9c1d8","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-03-09 19:55:54.000000000","message":"Patch Set 36:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/36 : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"eb006f2ba2b594adb9d357452660951e3f3dcb31","author":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"date":"2019-03-10 07:27:29.000000000","message":"Patch Set 36:\n\nBuild failed. To recheck use \u0027datacore: recheck\u0027. For information about DataCore CI see https://wiki.openstack.org/wiki/ThirdPartySystems/DataCore_CI\n\n- dsvm-tempest-datacore-iscsi-driver http://173.221.96.158/65/621465/36/check/dsvm-tempest-datacore-iscsi-driver/0bf668d : FAILURE in 51m 39s\n- dsvm-tempest-datacore-fc-driver http://173.221.96.158/65/621465/36/check/dsvm-tempest-datacore-fc-driver/17cbae0 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":36},{"id":"a4fc5a2d45463d4eb522114adcaf917f633f3507","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-03-10 08:55:10.000000000","message":"Patch Set 36:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/1518 : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"584c3d64df65e99168197c895bad7b34cd8ba82b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-03-10 09:58:31.000000000","message":"Patch Set 36:\n\nBuild succeed.\n\n- EMC_VMAX_FC http://publiclogs.emc.com/621465/36/EMC_VMAX_FC/954 : SUCCESS in 7h 13m 11s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/621465/36/EMC_VMAX_iSCSI/977 : SUCCESS in 7h 01m 20s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":36},{"id":"e95ecabe78f6a83400315f46be0811aa2888afd5","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-03-12 10:50:06.000000000","message":"Patch Set 36:\n\nrecheck intel_rsd_ci","accounts_in_message":[],"_revision_number":36},{"id":"b4245ea0817d40a0c147f7223506eb864693d915","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-12 11:33:00.000000000","message":"Patch Set 36:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/36/check-not-vote/ext-cinder-zuul/5550672 : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":36},{"id":"4b653d17afd5af9693d01242273fcf11efffc7bb","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-03-12 11:51:01.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/13078 : FAILURE in 36m 10s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9445 : FAILURE in 36m 09s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/11165 : SUCCESS in 1h 00m 32s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9820 : SUCCESS in 56m 08s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10792 : FAILURE in 37m 47s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1662 : FAILURE in 27s","accounts_in_message":[],"_revision_number":36},{"id":"cc150346a7e2a6e8dabc0fef6c0b28ee53dff27b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-03-12 12:12:41.000000000","message":"Patch Set 36:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/36/iscsi/ : FAILURE in 37m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/36/smb/ : SUCCESS in 1h 22m 10s","accounts_in_message":[],"_revision_number":36},{"id":"f7413d7c4cff1342d5aaafb508683d5d0b11d444","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-03-12 12:27:31.000000000","message":"Patch Set 36:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/36 : FAILURE","accounts_in_message":[],"_revision_number":36},{"id":"ac4bb40801226c31c6a94262331c66499549c4bc","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-03-12 12:34:28.000000000","message":"Patch Set 36:\n\nrecheck intel_rsd_ci","accounts_in_message":[],"_revision_number":36},{"id":"6b7698638ebc88cb8d6f1d7b4732cdd37ff629e9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-03-12 12:59:19.000000000","message":"Patch Set 36:\n\nMartin, please fix your recheck trigger. By starting with the word \"recheck\" you are causing Zuul to rerun check jobs and all other third party CI\u0027s that trigger off of zuul\u0027s +1.\n\nThe expected recheck trigger for third party CI is here:\n\nhttps://wiki.openstack.org/wiki/Cinder/tested-3rdParty-drivers#How_do_I_trigger_my_CI_to_rerun_on_gerrit_comments.3F","accounts_in_message":[],"_revision_number":36},{"id":"5806041ced213b037cac8ef7c10415b5ee0bdc85","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-03-12 13:10:11.000000000","message":"Patch Set 36:\n\n@Sean Fixed. Thank you for bringing this to my attention.","accounts_in_message":[],"_revision_number":36},{"id":"af0dff18c6ccb55ff8f60bd0988c8d05543bd2a3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-12 13:12:25.000000000","message":"Patch Set 36:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/36/check-not-vote/ext-cinder-zuul/edc8361 : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":36},{"id":"91688df449e9f8418bf9889e4e498aad1170e4e4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-03-12 13:26:06.000000000","message":"Patch Set 36:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/36/iscsi/ : FAILURE in 51m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/36/smb/ : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":36},{"id":"a681c543c7ab92d7c5c878511846c09fab8ad7d1","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-03-12 13:34:56.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/13079 : FAILURE in 37m 03s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9446 : FAILURE in 35m 42s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/11166 : SUCCESS in 1h 00m 08s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9821 : SUCCESS in 58m 25s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10793 : FAILURE in 37m 51s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1663 : FAILURE in 29s","accounts_in_message":[],"_revision_number":36},{"id":"2da37ae931be6ac84cc423b2445d8fd01de1458f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 13:40:47.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/36/check/openstack-tox-lower-constraints/247f266/ : SUCCESS in 9m 49s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/36/check/openstack-tox-pep8/9690581/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/36/check/openstack-tox-py27/e865b9e/ : SUCCESS in 8m 43s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/36/check/openstack-tox-py36/dd2ac46/ : SUCCESS in 8m 39s\n- openstack-tox-docs http://logs.openstack.org/65/621465/36/check/openstack-tox-docs/716f840/html/ : SUCCESS in 8m 12s\n- requirements-check http://logs.openstack.org/65/621465/36/check/requirements-check/e8b2c6b/ : SUCCESS in 3m 20s\n- tempest-full http://logs.openstack.org/65/621465/36/check/tempest-full/3a6f8bf/ : SUCCESS in 1h 43m 49s\n- neutron-grenade http://logs.openstack.org/65/621465/36/check/neutron-grenade/a92cd44/ : SUCCESS in 1h 03m 43s\n- grenade-py3 http://logs.openstack.org/65/621465/36/check/grenade-py3/31bbad3/ : SUCCESS in 53m 00s\n- tempest-full-py3 http://logs.openstack.org/65/621465/36/check/tempest-full-py3/03f2cd4/ : FAILURE in 1h 14m 28s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/36/check/build-openstack-releasenotes/7f7c48c/html/ : SUCCESS in 13m 21s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/36/check/cinder-tempest-dsvm-lvm-lio-barbican/c4b439f/ : SUCCESS in 1h 43m 32s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/36/check/cinder-tox-bandit-baseline/00c5b09/ : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/36/check/openstack-tox-functional-py36/dd01652/ : SUCCESS in 4m 43s\n- cinder-rally-task http://logs.openstack.org/65/621465/36/check/cinder-rally-task/c16ca98/ : SUCCESS in 40m 07s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/36/check/openstack-tox-pylint/585a368/ : SUCCESS in 5m 04s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/36/check/devstack-plugin-ceph-tempest/1957e5f/ : FAILURE in 1h 29m 45s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/36/check/openstack-tox-functional/6be14ca/ : SUCCESS in 5m 17s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/36/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/6d0dffd/ : SUCCESS in 1h 22m 28s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/36/check/legacy-tempest-dsvm-lvm-multibackend/8782afe/ : FAILURE in 1h 32m 09s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/36/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/cb871b8/ : SUCCESS in 1h 38m 37s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/36/check/tempest-slow-py3/019e574/ : SUCCESS in 2h 18m 23s","accounts_in_message":[],"_revision_number":36},{"id":"f742cd2ec916fc3a4dc4822256a9e3496b8dd872","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-03-12 14:15:03.000000000","message":"Patch Set 36:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/36 : FAILURE","accounts_in_message":[],"_revision_number":36},{"id":"c70f035a0cc95ceeb42b56b1d1f92514a6b339ec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-12 15:53:11.000000000","message":"Patch Set 36:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/36/check/tempest-dsvm-full-xenial/cf3295b/ : SUCCESS in 1h 00m 59s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/36/check/devstack-plugin-ceph-dsvm-xenial/64d0f89/ : SUCCESS in 1h 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"9272758b4cdf41d639ff792053162c3ddbe58842","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-03-12 16:03:21.000000000","message":"Patch Set 36:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":36},{"id":"4f0e51d488695b31f365e4bd8c0a07f7dce08ffc","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-03-12 17:38:28.000000000","message":"Patch Set 36:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/36 : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"47b49ac69500ccb3c76ea0dcd684062c90bbff86","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-03-12 18:10:23.000000000","message":"Patch Set 36: Code-Review-1\n\n(1 comment)\n\nThis driver should use the \u0027nvme\u0027 connector protocol in initialize_connection as it goes to the same os-brick connector as \u0027nvmeof\u0027\n\nThe \u0027nvmeof\u0027 connector protocol in os-brick should get reverted.","accounts_in_message":[],"_revision_number":36},{"id":"081757378f84f61baec9f681a2b8cb0108cf4dc8","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-03-12 19:41:05.000000000","message":"Patch Set 36: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"5f9559ea0afa8314459ab02a8644be79570440b8","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-03-12 21:01:43.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"14d17707b320e387c2a27114a7480a32a06ed73f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-12 21:43:19.000000000","message":"Patch Set 36:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/36/check-not-vote/ext-cinder-zuul/23f8b4e : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":36},{"id":"8a4846b447b65319ed9d5b6ec25162d22fdf96f1","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-03-12 21:58:50.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/36/check/dsvm-tempest-kvm/f2799ab : FAILURE in 49m 25s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/36/check/dsvm-tempest-exe/5f20994 : FAILURE in 49m 05s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":36},{"id":"b4583af3ae8990cd157bb66739e6377fc280965b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-03-12 21:59:08.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/36/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/21dc204 : SUCCESS in 52m 01s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/36/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/d322bd7 : SUCCESS in 57m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":36},{"id":"3dd62baf18c30ba0d7fa0272e1b679d681ed0772","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-03-12 22:01:45.000000000","message":"Patch Set 36:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/36/iscsi/ : FAILURE in 59m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/36/smb/ : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":36},{"id":"8687bb04948257ce19ef5e5e6ba7b7c5707713d7","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-03-12 22:04:09.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/13083 : FAILURE in 36m 47s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9450 : FAILURE in 36m 30s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/11170 : SUCCESS in 58m 46s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/9825 : SUCCESS in 1h 02m 06s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/10797 : FAILURE in 39m 42s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1667 : FAILURE in 36s","accounts_in_message":[],"_revision_number":36},{"id":"ba05e2f63c3fb5a4d80212e4d5c178552c13b8f2","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-03-12 23:14:39.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/36/check/Cinder-AS13000/7b4b1d5 : SUCCESS in 2h 12m 11s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/36/check/cinder-test-fc/4e2acba : SUCCESS in 56m 40s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/36/check/cinder-test-iscsi/5611503 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":36},{"id":"5a7efb04a36e04e7c3ffb2aeda2dde38d2ffb818","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-12 23:41:32.000000000","message":"Patch Set 36:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/36/check/tempest-dsvm-full-xenial/7ddd5f0/ : SUCCESS in 1h 08m 58s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/36/check/devstack-plugin-ceph-dsvm-xenial/c3e0843/ : SUCCESS in 1h 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"470c1b177002f2e6205e94caea3f0c3139e18802","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 01:15:42.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/36/check/openstack-tox-lower-constraints/e79a1f8/ : SUCCESS in 13m 11s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/36/check/openstack-tox-pep8/5fc4a2e/ : SUCCESS in 9m 07s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/36/check/openstack-tox-py27/3f60b91/ : SUCCESS in 11m 23s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/36/check/openstack-tox-py36/6f0a6bc/ : SUCCESS in 10m 31s\n- openstack-tox-docs http://logs.openstack.org/65/621465/36/check/openstack-tox-docs/2beee72/html/ : SUCCESS in 10m 22s\n- requirements-check http://logs.openstack.org/65/621465/36/check/requirements-check/2075665/ : SUCCESS in 5m 05s\n- tempest-full http://logs.openstack.org/65/621465/36/check/tempest-full/21e09c8/ : SUCCESS in 1h 40m 59s\n- neutron-grenade http://logs.openstack.org/65/621465/36/check/neutron-grenade/3a0db31/ : SUCCESS in 57m 55s\n- grenade-py3 http://logs.openstack.org/65/621465/36/check/grenade-py3/e9e092b/ : SUCCESS in 1h 05m 42s\n- tempest-full-py3 http://logs.openstack.org/65/621465/36/check/tempest-full-py3/423841f/ : SUCCESS in 1h 30m 22s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/36/check/build-openstack-releasenotes/2517bd2/html/ : SUCCESS in 15m 05s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/36/check/cinder-tempest-dsvm-lvm-lio-barbican/1491da0/ : SUCCESS in 1h 34m 38s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/36/check/cinder-tox-bandit-baseline/a555149/ : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/36/check/openstack-tox-functional-py36/f9cfb9c/ : SUCCESS in 7m 10s\n- cinder-rally-task http://logs.openstack.org/65/621465/36/check/cinder-rally-task/a5c5294/ : SUCCESS in 37m 26s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/36/check/openstack-tox-pylint/5e9a165/ : SUCCESS in 8m 05s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/36/check/devstack-plugin-ceph-tempest/d5c5255/ : SUCCESS in 1h 28m 26s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/36/check/openstack-tox-functional/17b825d/ : SUCCESS in 6m 50s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/36/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/32b4fe5/ : SUCCESS in 1h 04m 37s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/36/check/legacy-tempest-dsvm-lvm-multibackend/1f38e13/ : SUCCESS in 1h 35m 39s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/36/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/d64666b/ : SUCCESS in 1h 37m 59s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/36/check/tempest-slow-py3/8acf7d9/ : SUCCESS in 2h 10m 57s","accounts_in_message":[],"_revision_number":36},{"id":"1f8df281f97f2cccf18ce8dae0b86ba4b51f4d97","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-03-13 01:49:19.000000000","message":"Patch Set 36:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-36/ns5_iscsi : SUCCESS in 31m 28s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-36/ns5_nfs : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":36},{"id":"166d3241573cc7dff10c58cd2bac130754e30c4c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-03-13 01:52:06.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/36/upstream-solidfire-check/cinder-solidfire-iSCSI/26831c0/ : FAILURE in 35m 00s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":36},{"id":"44bcb0bba6bdd2e2613abafb4fdc7ab687f7f8c5","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-03-13 02:11:00.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/65/621465/36/check/synology-dsm-cinder-iscsi/4bcde34 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":36},{"id":"53ab9bdb13b7d8c8caefefc1cc03e0528a02b56f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-13 02:22:13.000000000","message":"Patch Set 36:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-36 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":36},{"id":"d0ec49443a068c3605a3a6a8df2f9760725e3ebf","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-03-13 02:34:52.000000000","message":"Patch Set 36:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-36 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI for rechecking and info.","accounts_in_message":[],"_revision_number":36},{"id":"66ce0cd2ca579a07acf6404cc9a57961fbdee3b3","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-03-13 02:36:04.000000000","message":"Patch Set 36:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-36-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"f49af7f1ca945711a6093f0cb2c405e8d70f6c6d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-03-13 03:04:40.000000000","message":"Patch Set 36:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/36/Lenovo-FC : SUCCESS in 38:00.46\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/36/Lenovo-iSCSI : SUCCESS in 38:02.72","accounts_in_message":[],"_revision_number":36},{"id":"1892be36e1488e5cd86e86ad221f0b21689c71c4","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-03-13 03:21:54.000000000","message":"Patch Set 36:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/36/HPMSA-FC : SUCCESS in 47:20.06\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/36/HPMSA-iSCSI : SUCCESS in 45:53.95","accounts_in_message":[],"_revision_number":36},{"id":"48287f3a1fe05b1f1d5be00095d1cf3c8a82217a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-03-13 03:24:00.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/36/upstream-check/cinder-cDOT-iSCSI/bb7f27b/ : FAILURE in 54m 53s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/36/upstream-check/cinder-cDOT-NFS/c379516/ : SUCCESS in 1h 05m 03s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/36/upstream-check/cinder-cDOT-FCP/5f23cdd/ : SUCCESS in 55m 07s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":36},{"id":"285f8696b30d81e5632707b856372c965205a28b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-03-13 03:42:16.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/36/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/16be9af : SUCCESS in 2h 12m 14s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/36/check/3par-fc-driver-master-client-pip-ssa02-dsvm/9b1535c : SUCCESS in 2h 23m 59s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/36/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/da7f687 : SUCCESS in 2h 11m 29s","accounts_in_message":[],"_revision_number":36},{"id":"d4790feda1d9b2b744a4ac0c46d654c9887b36cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-03-13 04:26:52.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/36/check/dsvm-tempest-cisco-zonemanager-job/03a74aa/ : FAILURE in 53m 55s","accounts_in_message":[],"_revision_number":36},{"id":"dd83a08a6332da4f3fdd5669a24723687b5b6764","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-03-13 04:51:33.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/36/check/kaminario-dsvm-tempest-full-iscsi/74bb410 : SUCCESS in 1h 52m 41s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/36/check/kaminario-dsvm-tempest-full-FC/450e4a3 : SUCCESS in 1h 46m 06s","accounts_in_message":[],"_revision_number":36},{"id":"671c4db4a209a31e3363d2f2f212469bfa874059","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-03-13 07:24:30.000000000","message":"Patch Set 36:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/36/check/EMC_XtremIO_FC/242c1b6/EMC_XtremIO_FC/1932 : SUCCESS in 1h 30m 00s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/36/check/EMC_XtremIO_iSCSI/2e7c803/EMC_XtremIO_iSCSI/1912 : SUCCESS in 1h 22m 56s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":36},{"id":"737011ebfa658c08755ad0cb9206e5736e8797a1","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-03-17 22:05:18.000000000","message":"Patch Set 36:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":36},{"id":"4ac3632ced21630ea5536a50aad5360de8ccb47c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-03-17 23:40:22.000000000","message":"Patch Set 36:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/36 : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"e8f5f7677e5fa1c8c14f27ae59ee423fe78515f1","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-03-29 12:02:28.000000000","message":"Patch Set 36:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/36 : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"b83f9c764bc21bd2cde7e2acc23e5d3e2fd74544","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-09 19:06:53.000000000","message":"Topic changed from bp/nvme-volume-driver to train-drivers","accounts_in_message":[],"_revision_number":36},{"id":"2a2ae7e4bf6fd0dddf65d7828a54ddaaaa33bf8d","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-09 23:37:58.000000000","message":"Patch Set 36:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":36},{"id":"981e70d2646be8353fc4897f35190a06b6169bee","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-10 18:34:37.000000000","message":"Patch Set 36:\n\nrecheck intel_rsd_ci","accounts_in_message":[],"_revision_number":36},{"id":"d573f1bb631391a2a7b5a2d493f0d95bfef7190d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-10 19:31:08.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/36/iscsi/ : SUCCESS in 55m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/36/smb/ : SUCCESS in 48m 30s","accounts_in_message":[],"_revision_number":36},{"id":"431fa2b498c4402076bbe8367b1a13ffbe3db667","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-10 20:08:01.000000000","message":"Patch Set 36:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/36/check/tempest-dsvm-full-xenial/38fcd2c/ : SUCCESS in 1h 14m 01s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/36/check/devstack-plugin-ceph-dsvm-xenial/98eb8da/ : SUCCESS in 1h 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"d589f44666206dc1d681db040a1db99df924104a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-10 20:24:37.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/36/check/dsvm-tempest-cisco-zonemanager-job/a0d917f/ : FAILURE in 1h 01m 46s","accounts_in_message":[],"_revision_number":36},{"id":"1a210c1cb3d8651160a7fe4cf5aabf324a6c693d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-10 20:25:00.000000000","message":"Patch Set 36:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/36/check-not-vote/ext-cinder-zuul/2c2f629 : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":36},{"id":"296fb49d7f03ba21cfa3c8e3607ae4ddcba27291","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 21:02:08.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/36/check/openstack-tox-lower-constraints/cbd05df/ : SUCCESS in 11m 29s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/36/check/openstack-tox-pep8/ffcd400/ : SUCCESS in 7m 31s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/36/check/openstack-tox-py27/0002d94/ : SUCCESS in 10m 56s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/36/check/openstack-tox-py36/7b6cc7d/ : SUCCESS in 10m 00s\n- openstack-tox-docs http://logs.openstack.org/65/621465/36/check/openstack-tox-docs/0ca9c94/html/ : SUCCESS in 7m 51s\n- requirements-check http://logs.openstack.org/65/621465/36/check/requirements-check/672c2e4/ : FAILURE in 3m 24s\n- tempest-full http://logs.openstack.org/65/621465/36/check/tempest-full/b2394ab/ : SUCCESS in 1h 49m 02s\n- neutron-grenade http://logs.openstack.org/65/621465/36/check/neutron-grenade/a98e9d3/ : SUCCESS in 47m 52s (non-voting)\n- grenade-py3 http://logs.openstack.org/65/621465/36/check/grenade-py3/2118978/ : SUCCESS in 55m 42s\n- tempest-full-py3 http://logs.openstack.org/65/621465/36/check/tempest-full-py3/864eb78/ : SUCCESS in 1h 34m 15s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/36/check/build-openstack-releasenotes/285896a/html/ : SUCCESS in 13m 53s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/36/check/cinder-tempest-dsvm-lvm-lio-barbican/d8d5f01/ : SUCCESS in 1h 40m 23s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/36/check/cinder-tox-bandit-baseline/ce5f227/ : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/36/check/openstack-tox-functional-py36/b0eaa21/ : SUCCESS in 5m 13s\n- cinder-rally-task http://logs.openstack.org/65/621465/36/check/cinder-rally-task/0521731/ : SUCCESS in 41m 28s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/36/check/openstack-tox-pylint/159c9cc/ : SUCCESS in 4m 39s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/36/check/devstack-plugin-ceph-tempest/6555890/ : SUCCESS in 1h 22m 52s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/36/check/openstack-tox-functional/49f501b/ : SUCCESS in 6m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/36/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/e1cba9f/ : SUCCESS in 1h 14m 28s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/36/check/legacy-tempest-dsvm-lvm-multibackend/6575529/ : SUCCESS in 1h 37m 26s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/36/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/40ccecf/ : SUCCESS in 1h 38m 26s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/36/check/tempest-slow-py3/9de3cd0/ : SUCCESS in 2h 22m 09s","accounts_in_message":[],"_revision_number":36},{"id":"66990c92151737ccfc33498176872efb8c08d91e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-11 15:25:11.000000000","message":"Patch Set 36:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/36 : FAILURE","accounts_in_message":[],"_revision_number":36},{"id":"3add34f899c0cea22599b7ecb193b149022ded36","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-11 18:01:54.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"1949c0688b89cbfb9f2cd7ab98c7933cf26ec606","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-04-11 18:02:42.000000000","message":"Patch Set 37:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":37},{"id":"fc6103caaa46c90920fbf7c0943796f2a71fefd2","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-11 18:06:11.000000000","message":"Patch Set 37:\n\nrecheck intel_rsd_ci","accounts_in_message":[],"_revision_number":37},{"id":"6ca89f61b89ff8ed2abb21dbe88ceaa3c205b9f4","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-04-11 18:07:18.000000000","message":"Patch Set 37:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":37},{"id":"628ec861af847334520b49c5f21074ae40346da7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-11 18:13:17.000000000","message":"Patch Set 37:\n\n\u003e recheck intel_rsd_ci\n\nAs mentioned previously, this third party CI recheck trigger needs to be changed. Starting with \"recheck\" causes all zuul jobs to run again, and all other third party CI that triggers on Zuul\u0027s +1.\n\nThe documented requirement for third party CI retriggering is \"run-Intel RSD CI\".\n\nhttps://wiki.openstack.org/wiki/Cinder/tested-3rdParty-drivers#How_do_I_trigger_my_CI_to_rerun_on_gerrit_comments.3F","accounts_in_message":[],"_revision_number":37},{"id":"b907bc8b4ad7319eceb591b90f18d131eeb5c02b","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-04-11 18:31:21.000000000","message":"Patch Set 37:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-04-11_11-02-00/ : FAILURE","accounts_in_message":[],"_revision_number":37},{"id":"9c6e7fe2240b75c1736e71bbfbd9582450e054df","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-11 18:38:55.000000000","message":"Patch Set 37:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/37/check-not-vote/ext-cinder-zuul/c0911c8 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":37},{"id":"4b89a462227a9516c4ce667a5f9d7cf83cc0fd3b","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-11 18:54:40.000000000","message":"Patch Set 37:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/37/check/dsvm-tempest-kvm/ebf17b7 : FAILURE in 50m 24s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/37/check/dsvm-tempest-exe/6a34856 : FAILURE in 50m 51s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":37},{"id":"26d1a9c1cbf3f04d166300d9283abe2d135e9d06","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-11 18:57:09.000000000","message":"Patch Set 37:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/37/iscsi/ : SUCCESS in 54m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/37/smb/ : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":37},{"id":"8965e9338a8e74094a420706e2d5473db64bab05","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-11 19:27:06.000000000","message":"Patch Set 37:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/37/check/dsvm-tempest-cisco-zonemanager-job/93bf7dd/ : FAILURE in 1h 13m 30s","accounts_in_message":[],"_revision_number":37},{"id":"f641e5169d36fd9e0f77a27b6e4d607809a41d2e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-11 19:30:48.000000000","message":"Patch Set 37:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/37/check/tempest-dsvm-full-xenial/3a646d5/ : SUCCESS in 1h 13m 38s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/37/check/devstack-plugin-ceph-dsvm-xenial/623b510/ : SUCCESS in 1h 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"0111e4dc2eb9d6e9e089faf9179a2e6e8c536e9f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-04-11 19:34:11.000000000","message":"Patch Set 37:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/37/check/fujitsu-eternus-dx-iscsi/59a21fd : FAILURE in 17m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/37/check/fujitsu-eternus-dx-fc/76ff29d : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":37},{"id":"f2fd556202a6d16e3bf3cdde77f039bd0c5619f4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-11 19:44:49.000000000","message":"Patch Set 37:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/37/check/dsvm-tempest-infinibox-fc/c62eadf : FAILURE in 58m 26s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/37/check/dsvm-tempest-infinibox-iscsi/4a91eeb : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":37},{"id":"17727d9ffb7389ec319cd11621db5b7b426d3f6b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-04-11 19:47:27.000000000","message":"Patch Set 37:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/37 : FAILURE","accounts_in_message":[],"_revision_number":37},{"id":"e577f8aab661c5d6e2691a0b9ebb1aa2faba8598","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-11 19:48:14.000000000","message":"Patch Set 37:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/37/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/6689bf5 : SUCCESS in 59m 43s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/37/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a54f759 : SUCCESS in 1h 10m 44s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":37},{"id":"7cf16eb7cb104d0dbaef59f59a33b196a01aac74","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-04-11 20:11:31.000000000","message":"Patch Set 37:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/37/check/NEC-ISCSI/084119b : SUCCESS in 1h 43m 50s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/37/check/NEC-FC/433e0a1 : SUCCESS in 1h 31m 20s","accounts_in_message":[],"_revision_number":37},{"id":"a17e6cfb0d44610a66bda70c10b70221ba43bcc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 20:23:44.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/37/check/openstack-tox-lower-constraints/736f543/ : SUCCESS in 11m 29s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/37/check/openstack-tox-pep8/8dd7c1e/ : SUCCESS in 9m 17s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/37/check/openstack-tox-py27/07292f6/ : SUCCESS in 11m 34s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/37/check/openstack-tox-py36/2d06781/ : SUCCESS in 10m 47s\n- openstack-tox-docs http://logs.openstack.org/65/621465/37/check/openstack-tox-docs/03aeddb/html/ : SUCCESS in 10m 45s\n- requirements-check http://logs.openstack.org/65/621465/37/check/requirements-check/7b513d2/ : FAILURE in 5m 00s\n- tempest-full http://logs.openstack.org/65/621465/37/check/tempest-full/88cfc25/ : SUCCESS in 1h 57m 27s\n- neutron-grenade http://logs.openstack.org/65/621465/37/check/neutron-grenade/d1d2cd1/ : SUCCESS in 1h 11m 05s\n- grenade-py3 http://logs.openstack.org/65/621465/37/check/grenade-py3/5c26712/ : SUCCESS in 53m 10s\n- tempest-full-py3 http://logs.openstack.org/65/621465/37/check/tempest-full-py3/1da1958/ : SUCCESS in 1h 44m 20s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/37/check/build-openstack-releasenotes/86959c4/html/ : SUCCESS in 16m 26s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/37/check/cinder-tempest-dsvm-lvm-lio-barbican/8d811f2/ : SUCCESS in 1h 29m 10s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/37/check/cinder-tox-bandit-baseline/27a3f34/ : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/37/check/openstack-tox-functional-py36/c8ba52b/ : SUCCESS in 7m 04s\n- cinder-rally-task http://logs.openstack.org/65/621465/37/check/cinder-rally-task/686adf6/ : SUCCESS in 37m 35s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/37/check/openstack-tox-pylint/ec712f6/ : SUCCESS in 6m 52s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/37/check/devstack-plugin-ceph-tempest/00e98d0/ : SUCCESS in 1h 28m 21s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/37/check/openstack-tox-functional/f617ede/ : SUCCESS in 7m 19s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/37/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/332e870/ : SUCCESS in 1h 03m 01s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/37/check/legacy-tempest-dsvm-lvm-multibackend/9098e74/ : SUCCESS in 1h 18m 15s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/37/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2cc5cb9/ : SUCCESS in 1h 29m 14s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/37/check/tempest-slow-py3/3902625/ : FAILURE in 2h 17m 11s","accounts_in_message":[],"_revision_number":37},{"id":"7dd91a197b8a57b0225ae4d9dd3bd606ae21a22c","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-04-11 21:13:14.000000000","message":"Patch Set 37:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":37},{"id":"7c1df5a801bb99598ab287c3fab687ef77427c11","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-11 22:46:43.000000000","message":"Patch Set 37:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/37 : SUCCESS","accounts_in_message":[],"_revision_number":37},{"id":"4135e7b88c2ecf6231ba7bec6580a7bf5dad8bd8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-04-11 23:18:17.000000000","message":"Patch Set 37:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/37/check/EMC_XtremIO_FC/6020e01/EMC_XtremIO_FC/2111 : SUCCESS in 50m 32s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/37/check/EMC_XtremIO_iSCSI/e27099c/EMC_XtremIO_iSCSI/2088 : SUCCESS in 1h 17m 51s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":37},{"id":"ceb5af31770aa4a818ea533c9252e591591c7c41","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-12 00:49:50.000000000","message":"Patch Set 37:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/37 : SUCCESS","accounts_in_message":[],"_revision_number":37},{"id":"829c4f132faa82790a54c1128da1eeab04e53b51","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-04-12 05:23:04.000000000","message":"Patch Set 38: Patch Set 37 was rebased","accounts_in_message":[],"_revision_number":38},{"id":"28727e077f0da4b57cd0f9da2b4322163aeb836a","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-12 05:37:21.000000000","message":"Topic changed from bp/nvme-volume-driver to train-drivers","accounts_in_message":[],"_revision_number":38},{"id":"91649bef60208c3c8866f1128103a5d7e13f96f5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 06:04:35.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/38/check/dsvm-tempest-kvm/22c38a9 : FAILURE in 40m 56s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/38/check/dsvm-tempest-exe/d01276a : FAILURE in 37m 01s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":38},{"id":"9b2bb429167ec5049faa86f488cdda56a2223a44","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-04-12 06:15:40.000000000","message":"Patch Set 38:\n\nBuild successful. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-04-11_22-23-12/ : SUCCESS","accounts_in_message":[],"_revision_number":38},{"id":"aaa64afc55131f54f60297cc92a22b1989017fdc","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-12 06:32:46.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/38/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/64417c7 : SUCCESS in 1h 03m 16s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/38/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/92557a3 : SUCCESS in 1h 09m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":38},{"id":"316a1ccc0dab2d2379c7d2f2b054bba8a4cf3bf9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-12 06:34:58.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/38/check/dsvm-tempest-infinibox-fc/1dea237 : SUCCESS in 1h 04m 33s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/38/check/dsvm-tempest-infinibox-iscsi/9dce136 : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":38},{"id":"8b6505a8019d573c078bebfd45a32e005c29de27","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-12 06:39:59.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/38/check/dsvm-tempest-cisco-zonemanager-job/045f440/ : FAILURE in 1h 05m 32s","accounts_in_message":[],"_revision_number":38},{"id":"f7097305aba75f8b1e6686c05d2fd5ee61a75793","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 06:41:56.000000000","message":"Patch Set 38:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/65/621465/38/check/EMC_XtremIO_FC/74eec55/EMC_XtremIO_FC/2116 : SUCCESS in 50m 05s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/38/check/EMC_XtremIO_iSCSI/fd76727/EMC_XtremIO_iSCSI/2093 : SUCCESS in 1h 18m 32s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":38},{"id":"4d5ec26965ffc9c84dbe5a5dd6e8910eb1172163","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-04-12 06:55:43.000000000","message":"Patch Set 38:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/38 : FAILURE","accounts_in_message":[],"_revision_number":38},{"id":"b47bdd53fb0d9c2e56eafe69798fe102f87ca33c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-12 06:56:01.000000000","message":"Patch Set 38:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/38 : SUCCESS","accounts_in_message":[],"_revision_number":38},{"id":"7152611328f46c826f6aed07117d09bd48a66d40","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-04-12 06:57:38.000000000","message":"Patch Set 38:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/38/check/fujitsu-eternus-dx-iscsi/748a84b : FAILURE in 17m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/38/check/fujitsu-eternus-dx-fc/1fccf98 : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":38},{"id":"2d114d027a60229be902b1c83d7afc8482111331","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 07:39:35.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/38/check/NEC-ISCSI/8b38fc0 : SUCCESS in 1h 59m 15s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/38/check/NEC-FC/3df017a : SUCCESS in 1h 53m 01s","accounts_in_message":[],"_revision_number":38},{"id":"dd5fe9edbf31c7d75c63d33613a949d98ca3bc00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 07:48:12.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/38/check/openstack-tox-lower-constraints/5e9acba/ : SUCCESS in 10m 34s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/38/check/openstack-tox-pep8/0795eb2/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/38/check/openstack-tox-py27/119b5c1/ : SUCCESS in 9m 08s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/38/check/openstack-tox-py36/4a8e3fa/ : SUCCESS in 8m 37s\n- openstack-tox-docs http://logs.openstack.org/65/621465/38/check/openstack-tox-docs/08e6d59/html/ : SUCCESS in 7m 52s\n- requirements-check http://logs.openstack.org/65/621465/38/check/requirements-check/90d544a/ : FAILURE in 4m 05s\n- tempest-full http://logs.openstack.org/65/621465/38/check/tempest-full/1d4b7a0/ : SUCCESS in 2h 05m 52s\n- neutron-grenade http://logs.openstack.org/65/621465/38/check/neutron-grenade/96bf969/ : SUCCESS in 55m 36s\n- grenade-py3 http://logs.openstack.org/65/621465/38/check/grenade-py3/ebe52a4/ : SUCCESS in 59m 27s\n- tempest-full-py3 http://logs.openstack.org/65/621465/38/check/tempest-full-py3/9252634/ : SUCCESS in 1h 49m 00s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/38/check/build-openstack-releasenotes/4f86cc4/html/ : SUCCESS in 13m 43s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/38/check/cinder-tempest-dsvm-lvm-lio-barbican/0b96312/ : SUCCESS in 1h 56m 10s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/38/check/cinder-tox-bandit-baseline/400886c/ : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/38/check/openstack-tox-functional-py36/6d75bf7/ : SUCCESS in 4m 43s\n- cinder-rally-task http://logs.openstack.org/65/621465/38/check/cinder-rally-task/7632fbc/ : SUCCESS in 47m 15s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/38/check/openstack-tox-pylint/cd2bd88/ : SUCCESS in 5m 36s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/38/check/devstack-plugin-ceph-tempest/95103f0/ : SUCCESS in 1h 33m 19s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/38/check/openstack-tox-functional/d075c60/ : SUCCESS in 6m 05s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/38/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/4e91a08/ : SUCCESS in 1h 08m 15s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/38/check/legacy-tempest-dsvm-lvm-multibackend/9b6a8f6/ : SUCCESS in 1h 32m 55s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/38/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/cb7dcfc/ : SUCCESS in 1h 28m 39s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/38/check/tempest-slow-py3/01e967a/ : SUCCESS in 2h 21m 42s","accounts_in_message":[],"_revision_number":38},{"id":"c983cc09afbd24b9139ec0aebb575078c4982543","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-12 07:59:00.000000000","message":"Patch Set 38:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/38/iscsi/ : FAILURE in 12m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/38/smb/ : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":38},{"id":"a8cf2330792dee18e054e97a945a5302c3dd08bf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-12 14:02:21.000000000","message":"Patch Set 38:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/38/check-not-vote/ext-cinder-zuul/6f8d838 : SUCCESS in 44m 35s","accounts_in_message":[],"_revision_number":38},{"id":"2c231cde9c4a20e0a670a5217a21742f946aa413","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-12 20:58:18.000000000","message":"Patch Set 38:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/38/check/tempest-dsvm-full-xenial/e805b5d/ : SUCCESS in 1h 21m 28s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/38/check/devstack-plugin-ceph-dsvm-xenial/279ca91/ : SUCCESS in 1h 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"f1bb966bb7a95c08d16850f44e24b1b1f9d5e4a1","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-04-13 07:48:17.000000000","message":"Patch Set 38:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_FC http://publiclogs.emc.com/621465/38/EMC_VMAX_FC/1023 : SUCCESS in 5h 53m 51s\n- EMC_VMAX_iSCSI http://publiclogs.emc.com/621465/38/EMC_VMAX_iSCSI/1053 : FAILURE in 7s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":38},{"id":"e18e72e41c6dfd0c9e9186cc8c4e0d486185b3dd","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-22 22:50:15.000000000","message":"Patch Set 38:\n\nrun-Zuul","accounts_in_message":[],"_revision_number":38},{"id":"2633c4990f27f04d4bded502874084d0344b43c6","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-23 15:25:23.000000000","message":"Patch Set 38:\n\nrun Zuul check","accounts_in_message":[],"_revision_number":38},{"id":"eca646d380aa5fdfb8b0eadc2b94a20818e801e1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-23 15:25:58.000000000","message":"Patch Set 38:\n\nrecheck","accounts_in_message":[],"_revision_number":38},{"id":"01a00b760ab2bb777af872d282ac09d7c09eaf73","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-23 15:26:26.000000000","message":"Patch Set 38:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/38/check/tempest-dsvm-full-xenial/4c9748e/ : FAILURE in 10s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/38/check/devstack-plugin-ceph-dsvm-xenial/b19450c/ : FAILURE in 10s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"a3faad0f6dd012a6893e0018c78cadcd4f0c3db0","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-23 15:27:39.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/38/check/dsvm-tempest-kvm/b79f5cd : FAILURE in 3s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/38/check/dsvm-tempest-exe/1ac0c20 : FAILURE in 4s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":38},{"id":"26de7d96c3b1d4a38dcb5b57fe68890e9d49d779","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-23 15:30:41.000000000","message":"Patch Set 38:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/38/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/887120c : FAILURE in 4m 16s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/38/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/93280c2 : FAILURE in 4m 25s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":38},{"id":"6dc06974ac9123e8d74730551357b1c99fecb178","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-23 16:00:07.000000000","message":"Patch Set 38:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/38/check-not-vote/ext-cinder-zuul/2db205e : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":38},{"id":"e952f1268e2f65243e4b8c046519529f1756bdab","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-23 16:19:05.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/38/check/dsvm-tempest-cisco-zonemanager-job/2b00e1a/ : FAILURE in 12m 19s","accounts_in_message":[],"_revision_number":38},{"id":"7d37c42b11edcba4d5680402e97f25770bfb26c0","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-23 16:20:57.000000000","message":"Patch Set 38:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/38/iscsi/ : FAILURE in 54m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/38/smb/ : SUCCESS in 47m 54s","accounts_in_message":[],"_revision_number":38},{"id":"063ba58536ec0524c6b357f6fdb9fd327e7ca82e","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-04-23 18:10:34.000000000","message":"Patch Set 38:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/38/check/Cinder-AS13000/3d54d1e : FAILURE in 15m 43s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/38/check/cinder-test-fc/6a2434a : SUCCESS in 58m 41s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/38/check/cinder-test-iscsi/dea28c5 : FAILURE in 14m 23s","accounts_in_message":[],"_revision_number":38},{"id":"c7ac47813ae578ca22a7d7331f09c2ce0c65167c","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-23 23:11:34.000000000","message":"Patch Set 38:\n\n\u003e recheck\n\nConfigured Intel RSD CI to be triggered by recheck, doing recheck again for verification.\n\n \u003e recheck","accounts_in_message":[],"_revision_number":38},{"id":"7fa054193e3eac653fb8a542d01ed7f1a14940f4","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-23 23:14:22.000000000","message":"Patch Set 38:\n\nrecheck","accounts_in_message":[],"_revision_number":38},{"id":"e65c5f48ef05e6af13865ca5ef6ff483be9f9d69","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-23 23:14:53.000000000","message":"Patch Set 38:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/38/check/tempest-dsvm-full-xenial/0f9f222/ : FAILURE in 10s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/38/check/devstack-plugin-ceph-dsvm-xenial/7288583/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"90a5780c625206e7b84f34388008f6b39e45004a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-23 23:15:58.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/38/check/dsvm-tempest-kvm/7bd9499 : FAILURE in 2s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/38/check/dsvm-tempest-exe/9dc2c78 : FAILURE in 28s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":38},{"id":"23622a63285fe7e07cb751cb17e2507e1ea34f16","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-23 23:19:11.000000000","message":"Patch Set 38:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/38/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/b11048c : FAILURE in 4m 23s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/38/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a5cb5eb : FAILURE in 4m 26s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":38},{"id":"8c6ab47fb9d83b687149a256ce98b81f6a5b1788","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-23 23:46:27.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/38/check/dsvm-tempest-cisco-zonemanager-job/c5aaead/ : FAILURE in 13m 16s","accounts_in_message":[],"_revision_number":38},{"id":"ee20d77fe1a966feb1c053633360e4af9a122384","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-23 23:54:03.000000000","message":"Patch Set 38:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/38/check-not-vote/ext-cinder-zuul/0290520 : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":38},{"id":"b83a4e880adcd5af3ed36ba4309b68f78baf5c4f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-24 00:07:01.000000000","message":"Patch Set 38:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/38/iscsi/ : FAILURE in 52m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/38/smb/ : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":38},{"id":"2847e1bcc181d77671163a804502a98d0853206d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-24 00:47:07.000000000","message":"Patch Set 38:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/38 : SUCCESS","accounts_in_message":[],"_revision_number":38},{"id":"dd9f892e9cc9fd15d1ced4e86cc77ecccbe2ed41","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-04-24 01:43:57.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/38/check/Cinder-AS13000/99d0af2 : SUCCESS in 2h 28m 14s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/38/check/cinder-test-fc/3c09b64 : SUCCESS in 51m 55s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/38/check/cinder-test-iscsi/edbee96 : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":38},{"id":"cab254ffe701e7f5ab5b7f87a4c0ebe6a0474cf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 03:51:42.000000000","message":"Patch Set 38:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/38/check/openstack-tox-lower-constraints/948f2f5/ : SUCCESS in 10m 05s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/38/check/openstack-tox-pep8/9a4ca6f/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/38/check/openstack-tox-py27/97d1efe/ : SUCCESS in 8m 48s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/38/check/openstack-tox-py36/6e9f52c/ : SUCCESS in 8m 09s\n- openstack-tox-docs http://logs.openstack.org/65/621465/38/check/openstack-tox-docs/17e7a77/html/ : SUCCESS in 8m 48s\n- requirements-check http://logs.openstack.org/65/621465/38/check/requirements-check/7be1e4f/ : SUCCESS in 2m 49s\n- tempest-full http://logs.openstack.org/65/621465/38/check/tempest-full/eb88233/ : POST_FAILURE in 1h 40m 42s\n- neutron-grenade http://logs.openstack.org/65/621465/38/check/neutron-grenade/c988013/ : SUCCESS in 56m 23s\n- grenade-py3 http://logs.openstack.org/65/621465/38/check/grenade-py3/d93223c/ : SUCCESS in 1h 05m 54s\n- tempest-full-py3 http://logs.openstack.org/65/621465/38/check/tempest-full-py3/345027d/ : SUCCESS in 1h 29m 08s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/38/check/build-openstack-releasenotes/48b56c6/html/ : SUCCESS in 16m 21s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/38/check/cinder-tempest-dsvm-lvm-lio-barbican/bdb4b44/ : SUCCESS in 1h 32m 17s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/38/check/cinder-tox-bandit-baseline/1de39ee/ : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/38/check/openstack-tox-functional-py36/199e820/ : SUCCESS in 7m 05s\n- cinder-rally-task http://logs.openstack.org/65/621465/38/check/cinder-rally-task/1c58500/ : SUCCESS in 42m 17s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/38/check/openstack-tox-pylint/d31dffa/ : SUCCESS in 5m 19s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/38/check/devstack-plugin-ceph-tempest/30d6c5f/ : SUCCESS in 1h 44m 05s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/38/check/openstack-tox-functional/5ae742c/ : SUCCESS in 5m 54s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/38/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/58c51bf/ : SUCCESS in 1h 07m 09s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/38/check/legacy-tempest-dsvm-lvm-multibackend/e46154d/ : SUCCESS in 1h 52m 51s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/38/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/27d18ea/ : SUCCESS in 1h 25m 22s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/38/check/tempest-slow-py3/0782399/ : SUCCESS in 2h 15m 21s","accounts_in_message":[],"_revision_number":38},{"id":"60b3ebf01fe4957b3bd9e84fd18f207e8e31f986","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-04-24 14:32:59.000000000","message":"Patch Set 38: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":38},{"id":"51796fef9f683ab590cd162a6e1668e7f0d8f9d7","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-24 15:54:28.000000000","message":"Patch Set 38:\n\nrecheck","accounts_in_message":[],"_revision_number":38},{"id":"f05c0665fc4566799ca4e8b684055ae8a78d3c08","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-24 15:55:31.000000000","message":"Patch Set 38:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/38/check/tempest-dsvm-full-xenial/b529d7a/ : FAILURE in 9s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/38/check/devstack-plugin-ceph-dsvm-xenial/2642b74/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"1f2583a99d6f89a1dfc53f057c976c5386869c6b","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 15:55:56.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/38/check/dsvm-tempest-kvm/f98cb34 : FAILURE in 3s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/38/check/dsvm-tempest-exe/a748bab : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":38},{"id":"38ba674d2c9272ccb1119e9aa6c995563dc77687","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-24 15:58:44.000000000","message":"Patch Set 38:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/38/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/869d176 : FAILURE in 3m 39s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/38/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/c1f6f7e : FAILURE in 3m 58s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":38},{"id":"822e3b4ef21cb894acce4a624c69ed7a054f294f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-24 16:20:38.000000000","message":"Patch Set 38:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/38/check-not-vote/ext-cinder-zuul/4a7eb6f : FAILURE in 25m 41s","accounts_in_message":[],"_revision_number":38},{"id":"c272b5ee9a31ff564965f05ee03df5735dad0290","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-24 16:33:21.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/38/check/dsvm-tempest-cisco-zonemanager-job/d3465ae/ : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":38},{"id":"dcdf2da798b5a51d843b72ea855b3fcc8816bcb3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-24 17:01:01.000000000","message":"Patch Set 38:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/38/iscsi/ : SUCCESS in 1h 06m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/38/smb/ : FAILURE in 56m 15s","accounts_in_message":[],"_revision_number":38},{"id":"870a68930042ebf803ecd9186be60fb337f29e6d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-24 17:28:35.000000000","message":"Patch Set 38:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/38 : SUCCESS","accounts_in_message":[],"_revision_number":38},{"id":"10d5930d490f408e49d71a162920a79268c09ead","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-04-24 18:52:00.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/38/check/Cinder-AS13000/6817802 : SUCCESS in 2h 40m 57s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/38/check/cinder-test-fc/4a21c97 : SUCCESS in 1h 03m 32s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/38/check/cinder-test-iscsi/caa466b : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":38},{"id":"577dbb45ce1c18a339021afdf5463ea787900697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 04:26:06.000000000","message":"Patch Set 38:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/38/check/openstack-tox-lower-constraints/11023b8/ : SUCCESS in 10m 11s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/38/check/openstack-tox-pep8/7c94bfb/ : SUCCESS in 7m 46s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/38/check/openstack-tox-py27/4fcfaac/ : SUCCESS in 9m 30s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/38/check/openstack-tox-py36/f9cacac/ : SUCCESS in 9m 25s\n- openstack-tox-docs http://logs.openstack.org/65/621465/38/check/openstack-tox-docs/a1eedd3/html/ : SUCCESS in 8m 41s\n- requirements-check http://logs.openstack.org/65/621465/38/check/requirements-check/4fd4811/ : SUCCESS in 2m 53s\n- tempest-full http://logs.openstack.org/65/621465/38/check/tempest-full/733dd48/ : TIMED_OUT in 2h 07m 18s\n- neutron-grenade http://logs.openstack.org/65/621465/38/check/neutron-grenade/0214fe4/ : SUCCESS in 56m 51s\n- grenade-py3 http://logs.openstack.org/65/621465/38/check/grenade-py3/d1421a3/ : SUCCESS in 59m 51s\n- tempest-full-py3 http://logs.openstack.org/65/621465/38/check/tempest-full-py3/7ac54e9/ : SUCCESS in 1h 23m 39s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/38/check/build-openstack-releasenotes/a469249/html/ : SUCCESS in 16m 50s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/38/check/cinder-tempest-dsvm-lvm-lio-barbican/041c90c/ : SUCCESS in 1h 34m 55s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/38/check/cinder-tox-bandit-baseline/313e498/ : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/38/check/openstack-tox-functional-py36/b0924b5/ : SUCCESS in 5m 14s\n- cinder-rally-task http://logs.openstack.org/65/621465/38/check/cinder-rally-task/448dbfa/ : SUCCESS in 36m 41s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/38/check/openstack-tox-pylint/50331e6/ : SUCCESS in 5m 29s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/38/check/devstack-plugin-ceph-tempest/59da1b7/ : SUCCESS in 1h 52m 58s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/38/check/openstack-tox-functional/f914ba9/ : SUCCESS in 5m 21s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/38/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/b614c35/ : SUCCESS in 1h 04m 24s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/38/check/legacy-tempest-dsvm-lvm-multibackend/47ad8d4/ : SUCCESS in 1h 25m 27s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/38/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/e4848ef/ : SUCCESS in 1h 25m 51s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/38/check/tempest-slow-py3/cfdce5e/ : SUCCESS in 2h 15m 20s","accounts_in_message":[],"_revision_number":38},{"id":"0b0dbe64299d6398541e56155a9f4f7927ce47e1","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-25 23:38:46.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"bafff8ae8c18b3b21cf3535580bda119e72e2e3b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-25 23:39:17.000000000","message":"Patch Set 39:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/39/check/tempest-dsvm-full-xenial/56152dd/ : FAILURE in 9s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/39/check/devstack-plugin-ceph-dsvm-xenial/0eaaa95/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"6275d4169b52cb879780dce0ee827eb912b299ee","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-04-25 23:39:19.000000000","message":"Patch Set 39:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/39/check/NEC-ISCSI/3c0d447 : FAILURE in 3s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/39/check/NEC-FC/f297667 : FAILURE in 3s","accounts_in_message":[],"_revision_number":39},{"id":"117b6e4134c69487a4953cdb5be65bfbcbc8f9a7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-04-25 23:39:20.000000000","message":"Patch Set 39:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/39 : FAILURE","accounts_in_message":[],"_revision_number":39},{"id":"8ff6bca77a199ee9a31527dbe88983a76bd716c4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-25 23:40:04.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/39/check/dsvm-tempest-kvm/929c429 : FAILURE in 2s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/39/check/dsvm-tempest-exe/e5c4af7 : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":39},{"id":"f288c7172cecf8eea1b94d9dafa065226c5ea8eb","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-25 23:43:37.000000000","message":"Patch Set 39:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/39/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/d3536ec : FAILURE in 4m 28s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/39/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/137db79 : FAILURE in 4m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":39},{"id":"e9c38d2111a4f0ef50cd5db390ec626ab3333a51","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-26 00:09:11.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/39/check/dsvm-tempest-cisco-zonemanager-job/903fedc/ : FAILURE in 11m 50s","accounts_in_message":[],"_revision_number":39},{"id":"17ee6663b8c5e11b06402b2e1c11b613f2579d4c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-26 00:13:52.000000000","message":"Patch Set 39:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/39/check-not-vote/ext-cinder-zuul/b70b330 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":39},{"id":"b09a1580f944617921c7abb3a49ebf45589727d8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-26 00:33:07.000000000","message":"Patch Set 39:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/39/iscsi/ : FAILURE in 53m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/39/smb/ : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":39},{"id":"d4671e2679b890d40905c11c51378f220696b6b9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-04-26 00:33:25.000000000","message":"Patch Set 39:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/39/check/DellEMC_XtremIO_FC/6e34160/DellEMC_XtremIO_FC/60 : SUCCESS in 54m 17s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/39/check/DellEMC_XtremIO_iSCSI/fb94270/DellEMC_XtremIO_iSCSI/59 : FAILURE in 2m 54s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":39},{"id":"5bc54d1ba1b0f72bc3a2cf883b8c899df7ede132","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-04-26 00:34:49.000000000","message":"Patch Set 39:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/39/check/DellEMC_Unity_FC/5de378b/DellEMC_Unity_FC/37 : SUCCESS in 55m 49s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/39/check/DellEMC_Unity_iSCSI/f91439e/DellEMC_Unity_iSCSI/37 : FAILURE in 21m 38s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":39},{"id":"03092dc1d68608544cd15806fb6267f80707345b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-26 00:51:10.000000000","message":"Patch Set 39:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/39/check/dsvm-tempest-infinibox-fc/2b3c140 : FAILURE in 14m 31s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/39/check/dsvm-tempest-infinibox-iscsi/4cfda09 : SUCCESS in 1h 11m 35s","accounts_in_message":[],"_revision_number":39},{"id":"2f0c5b5d01685e8fd146f77553af2459aeaa0674","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-04-26 01:01:21.000000000","message":"Patch Set 39:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/39/check/fujitsu-eternus-dx-iscsi/ca3c7b4 : FAILURE in 11m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/39/check/fujitsu-eternus-dx-fc/b6c20af : SUCCESS in 1h 21m 45s","accounts_in_message":[],"_revision_number":39},{"id":"a7c8a2eaa323de7d4bf8d2a2698c8d26db2658c0","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-04-26 01:16:57.000000000","message":"Patch Set 39:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/65/621465/39/check/Cinder-AS13000/4a4a09a : FAILURE in 19m 19s\n- cinder-test-fc http://218.29.111.78:9000/65/621465/39/check/cinder-test-fc/f2889b9 : SUCCESS in 56m 46s\n- cinder-test-iscsi http://218.29.111.78:9000/65/621465/39/check/cinder-test-iscsi/5fc4981 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":39},{"id":"5a8431e02d8307b1aa5c33a857e1e3b23505477a","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-26 02:21:32.000000000","message":"Topic changed from bp/nvme-volume-driver to train-drivers","accounts_in_message":[],"_revision_number":39},{"id":"8e6f50d5360e35ae8a07e72e1b5067612d2aa8e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 08:24:00.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/39/check/openstack-tox-lower-constraints/eb4a05b/ : SUCCESS in 10m 18s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/39/check/openstack-tox-pep8/251e45b/ : SUCCESS in 7m 08s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/39/check/openstack-tox-py27/67fc9f8/ : SUCCESS in 10m 19s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/39/check/openstack-tox-py36/52a6dcc/ : SUCCESS in 8m 26s\n- openstack-tox-docs http://logs.openstack.org/65/621465/39/check/openstack-tox-docs/04b4c5b/html/ : SUCCESS in 7m 42s\n- requirements-check http://logs.openstack.org/65/621465/39/check/requirements-check/e5c1df6/ : SUCCESS in 2m 36s\n- tempest-full http://logs.openstack.org/65/621465/39/check/tempest-full/6bf9397/ : SUCCESS in 1h 41m 11s\n- neutron-grenade http://logs.openstack.org/65/621465/39/check/neutron-grenade/aecbe98/ : SUCCESS in 56m 19s\n- grenade-py3 http://logs.openstack.org/65/621465/39/check/grenade-py3/71c7024/ : SUCCESS in 1h 03m 39s\n- tempest-full-py3 http://logs.openstack.org/65/621465/39/check/tempest-full-py3/0b58420/ : FAILURE in 24m 57s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/39/check/build-openstack-releasenotes/793e1ab/html/ : SUCCESS in 13m 40s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/39/check/cinder-tempest-dsvm-lvm-lio-barbican/6c42dd2/ : SUCCESS in 1h 45m 53s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/39/check/cinder-tox-bandit-baseline/3ec107f/ : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/39/check/openstack-tox-functional-py36/0679074/ : SUCCESS in 5m 00s\n- cinder-rally-task http://logs.openstack.org/65/621465/39/check/cinder-rally-task/95ee96e/ : SUCCESS in 38m 08s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/39/check/openstack-tox-pylint/8bcb192/ : SUCCESS in 4m 48s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/39/check/devstack-plugin-ceph-tempest/0530222/ : SUCCESS in 1h 28m 02s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/39/check/openstack-tox-functional/9ec488c/ : SUCCESS in 5m 34s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/39/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/44eb766/ : SUCCESS in 1h 09m 42s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/39/check/legacy-tempest-dsvm-lvm-multibackend/0122683/ : SUCCESS in 1h 29m 12s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/39/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/f960852/ : SUCCESS in 1h 39m 40s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/39/check/tempest-slow-py3/bedb418/ : SUCCESS in 2h 25m 40s","accounts_in_message":[],"_revision_number":39},{"id":"d8b2050f2197e994b1a47f9ffcb111541ce4626f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-04-26 10:16:14.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/39/check/dsvm-tempest-storpool/3d97388 : SUCCESS in 1h 59m 41s","accounts_in_message":[],"_revision_number":39},{"id":"326ce712091778d828104f9d23306e7caf471772","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-04-26 10:17:19.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/39/check-cinder/Cinder-ISER-ISCSI/a888720 : SUCCESS in 1h 09m 50s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/39/check-cinder/Cinder-ISER-LIO/8024686 : SUCCESS in 1h 11m 13s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":39},{"id":"2b3f16a58fb611e6b180c7db2eb1f6241d6224ab","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-26 10:55:02.000000000","message":"Patch Set 39:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/39 : FAILURE","accounts_in_message":[],"_revision_number":39},{"id":"5202bfc3b87bc5519a8417f4e87406d138233e72","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-04-26 16:07:17.000000000","message":"Patch Set 39:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":39},{"id":"a22477e95c8d085aa54c81d84f4c81ff66aae66a","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-26 17:45:07.000000000","message":"Patch Set 39:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/39 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"b3e7fa3dbc359d144c0c4e966431670bda412611","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-28 20:28:06.000000000","message":"Patch Set 39:\n\n(3 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"3a110d9a6e581571412bc4ad08853c4da9f37525","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-29 00:24:39.000000000","message":"Patch Set 39:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/39 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"0c003f02eb0bd0747883dbe0c8618d2170e7273e","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-29 19:02:31.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"256bf6292471c6eed085e447f7164cf11953a650","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 19:03:48.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/39/check/dsvm-tempest-kvm/067e207 : FAILURE in 3s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/39/check/dsvm-tempest-exe/2fe1507 : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":39},{"id":"f2e684255c90f18b292757fd4ce6add93903f825","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-29 19:07:52.000000000","message":"Patch Set 39:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/39/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/4315b07 : FAILURE in 4m 21s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/39/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/5e245bf : FAILURE in 5m 03s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":39},{"id":"4357ce25d9dee117a4a58b4961cb605592eecafa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-29 19:25:17.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/39/check/dsvm-tempest-cisco-zonemanager-job/116065c/ : FAILURE in 12m 08s","accounts_in_message":[],"_revision_number":39},{"id":"cfe4c5e483b84398ab969d5b6689f112756bba03","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-29 19:37:56.000000000","message":"Patch Set 39:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/39/check-not-vote/ext-cinder-zuul/a68bdbe : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":39},{"id":"cb8d5a52aa191cc294d23351c69c35c731ee91c5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-29 19:54:13.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/39/iscsi/ : SUCCESS in 51m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/39/smb/ : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":39},{"id":"07d3d12b080a066339bad37768c850937740c04d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-29 20:24:51.000000000","message":"Patch Set 39:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/39/check/tempest-dsvm-full-xenial/58e4c60/ : SUCCESS in 1h 13m 07s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/39/check/devstack-plugin-ceph-dsvm-xenial/6caa9ff/ : SUCCESS in 1h 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"b895920360a0f53e229096bad13bd2e5a969f60b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-29 20:40:10.000000000","message":"Patch Set 39:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/39 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"3f9f4be30d6421827b3da4989b2dfea44f6b931e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 22:07:32.000000000","message":"Patch Set 39:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/39/check/openstack-tox-lower-constraints/30714b6/ : SUCCESS in 9m 34s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/39/check/openstack-tox-pep8/63e9c73/ : SUCCESS in 9m 39s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/39/check/openstack-tox-py27/35898a7/ : SUCCESS in 11m 05s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/39/check/openstack-tox-py36/739e6c4/ : SUCCESS in 8m 52s\n- openstack-tox-docs http://logs.openstack.org/65/621465/39/check/openstack-tox-docs/ce03274/html/ : SUCCESS in 7m 48s\n- requirements-check http://logs.openstack.org/65/621465/39/check/requirements-check/df28878/ : SUCCESS in 3m 04s\n- tempest-full http://logs.openstack.org/65/621465/39/check/tempest-full/ac01c52/ : SUCCESS in 1h 53m 45s\n- neutron-grenade http://logs.openstack.org/65/621465/39/check/neutron-grenade/c030e63/ : SUCCESS in 58m 20s\n- grenade-py3 http://logs.openstack.org/65/621465/39/check/grenade-py3/4a310ce/ : SUCCESS in 1h 02m 56s\n- tempest-full-py3 http://logs.openstack.org/65/621465/39/check/tempest-full-py3/4e53075/ : SUCCESS in 1h 25m 56s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/39/check/build-openstack-releasenotes/677bd31/html/ : SUCCESS in 13m 45s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/39/check/cinder-tempest-dsvm-lvm-lio-barbican/2910aa0/ : FAILURE in 1h 45m 22s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/39/check/cinder-tox-bandit-baseline/d7943ce/ : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/39/check/openstack-tox-functional-py36/32a1843/ : SUCCESS in 4m 53s\n- cinder-rally-task http://logs.openstack.org/65/621465/39/check/cinder-rally-task/48b6fad/ : SUCCESS in 38m 58s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/39/check/openstack-tox-pylint/de54f46/ : SUCCESS in 4m 22s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/39/check/devstack-plugin-ceph-tempest/387848b/ : FAILURE in 1h 35m 48s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/39/check/openstack-tox-functional/5d7ee88/ : SUCCESS in 6m 48s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/39/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/5a0f671/ : SUCCESS in 1h 07m 54s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/39/check/legacy-tempest-dsvm-lvm-multibackend/b504053/ : SUCCESS in 1h 20m 36s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/39/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/d537c15/ : SUCCESS in 1h 52m 45s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/39/check/tempest-slow-py3/3d797b6/ : SUCCESS in 2h 15m 19s","accounts_in_message":[],"_revision_number":39},{"id":"0cd4810184fb8aa7a9c3c69f70df3c0189a22fcd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-30 13:13:14.000000000","message":"Patch Set 39:\n\nrecheck\n\nLooks like an unrelated network issue caused an SSH timeout to the instance.","accounts_in_message":[],"_revision_number":39},{"id":"b6fdc7f10597b101ce8343cf23d70772ed4396e9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-04-30 13:18:37.000000000","message":"Patch Set 39:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/39/check/dsvm-tempest-zadara-driver/6c7bc0f : FAILURE in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"113ca46e3a324013cbd981b28900d3349bce9e26","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-04-30 13:31:42.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/39/check/dsvm-tempest-cisco-zonemanager-job/22e73e9/ : FAILURE in 12m 02s","accounts_in_message":[],"_revision_number":39},{"id":"7eb66ab88d190ba88d257c5e86009217718039d4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-30 13:48:58.000000000","message":"Patch Set 39:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/39/check-not-vote/ext-cinder-zuul/bea1687 : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":39},{"id":"dc5382fba68978156d7b899ed77fd0ce8fc54d5b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-04-30 14:06:17.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/39/iscsi/ : SUCCESS in 52m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/39/smb/ : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":39},{"id":"99a67f8be02457211e139aea4912e51c668b118d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-30 14:35:53.000000000","message":"Patch Set 39:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/39/check/tempest-dsvm-full-xenial/7a19dfa/ : SUCCESS in 1h 21m 44s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/39/check/devstack-plugin-ceph-dsvm-xenial/42b86e7/ : SUCCESS in 1h 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"9a709019791447aa09d0bfd5eee23cff02886cb3","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-30 14:47:54.000000000","message":"Patch Set 39:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/39 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"a37d76263833488063eee12284bb99784a9a57ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 15:39:14.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/39/check/openstack-tox-lower-constraints/e5c957d/ : SUCCESS in 10m 50s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/39/check/openstack-tox-pep8/a2955f7/ : SUCCESS in 8m 57s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/39/check/openstack-tox-py27/4fee286/ : SUCCESS in 12m 32s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/39/check/openstack-tox-py36/f5a756c/ : SUCCESS in 8m 37s\n- openstack-tox-docs http://logs.openstack.org/65/621465/39/check/openstack-tox-docs/ae08067/html/ : SUCCESS in 7m 47s\n- requirements-check http://logs.openstack.org/65/621465/39/check/requirements-check/0bad8cf/ : SUCCESS in 4m 24s\n- tempest-full http://logs.openstack.org/65/621465/39/check/tempest-full/24a8b04/ : SUCCESS in 2h 04m 23s\n- neutron-grenade http://logs.openstack.org/65/621465/39/check/neutron-grenade/fa785d9/ : SUCCESS in 1h 11m 01s\n- grenade-py3 http://logs.openstack.org/65/621465/39/check/grenade-py3/c02b814/ : SUCCESS in 58m 32s\n- tempest-full-py3 http://logs.openstack.org/65/621465/39/check/tempest-full-py3/a467dad/ : SUCCESS in 1h 51m 52s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/39/check/build-openstack-releasenotes/be60f53/html/ : SUCCESS in 13m 22s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/39/check/cinder-tempest-dsvm-lvm-lio-barbican/fa0d4ae/ : SUCCESS in 1h 50m 05s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/39/check/cinder-tox-bandit-baseline/51c412f/ : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/39/check/openstack-tox-functional-py36/986933d/ : SUCCESS in 4m 50s\n- cinder-rally-task http://logs.openstack.org/65/621465/39/check/cinder-rally-task/f533490/ : SUCCESS in 36m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/39/check/openstack-tox-pylint/6bcd431/ : SUCCESS in 4m 42s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/39/check/devstack-plugin-ceph-tempest/f51caf8/ : SUCCESS in 1h 27m 01s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/39/check/openstack-tox-functional/a5dfb5e/ : SUCCESS in 5m 17s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/39/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/33dc6c4/ : SUCCESS in 1h 11m 39s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/39/check/legacy-tempest-dsvm-lvm-multibackend/3094bc3/ : SUCCESS in 1h 27m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/39/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/bce5bda/ : FAILURE in 1h 18m 17s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/39/check/tempest-slow-py3/9f38622/ : SUCCESS in 2h 20m 33s","accounts_in_message":[],"_revision_number":39},{"id":"0dea1744bba1510c1e726a3e3e991679966d3f19","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 15:40:01.000000000","message":"Patch Set 39:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/65/621465/39/check/DellEMC_VxFlexOS/e8c9364/DellEMC_VxFlexOS/13 : FAILURE in 29s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":39},{"id":"bd8f1f101c3a5287541539f6d763732d5aa47689","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-04-30 15:52:52.000000000","message":"Patch Set 39:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-39/nedge_iscsi : FAILURE in 11m 31s","accounts_in_message":[],"_revision_number":39},{"id":"41c98d858a14fa4b74e7081a8344073c703d69f1","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-04-30 15:56:48.000000000","message":"Patch Set 39:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/1873 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"5813c0645a4a4d7e2916606f8848613aec815193","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-04-30 16:03:06.000000000","message":"Patch Set 39:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/65/621465/39/check/synology-dsm-cinder-iscsi/52aa296 : FAILURE in 23m 21s","accounts_in_message":[],"_revision_number":39},{"id":"88b56410e922e60e743a6c18b29566bc8f2393f2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-04-30 16:04:26.000000000","message":"Patch Set 39:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-39/ns5_iscsi : FAILURE in 11m 39s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-39/ns5_nfs : FAILURE in 23m 7s","accounts_in_message":[],"_revision_number":39},{"id":"acc84dbc7e4ea4dee0d145a52c6997a82bd4057e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-30 16:51:40.000000000","message":"Patch Set 39:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-39 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":39},{"id":"f2209a06254dd04854cb7caa3d97aa76870ac95a","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-04-30 16:52:10.000000000","message":"Patch Set 39:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/1874 : FAILURE","accounts_in_message":[],"_revision_number":39},{"id":"29d916fce210a235485f47a4695ac807b1a26513","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-04-30 16:53:39.000000000","message":"Patch Set 39:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-39 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI","accounts_in_message":[],"_revision_number":39},{"id":"b97270cff5eef054367a4d2a1a635b7928cb31d0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-04-30 16:53:58.000000000","message":"Patch Set 39:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/39/check/DellEMC_Unity_FC/b3e2adc/DellEMC_Unity_FC/67 : SUCCESS in 1h 09m 24s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/39/check/DellEMC_Unity_iSCSI/f8d701d/DellEMC_Unity_iSCSI/67 : SUCCESS in 1h 14m 31s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":39},{"id":"9f79453b3989dd7d5cfcf458da9b88791b0fdf55","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 16:57:44.000000000","message":"Patch Set 39:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-39-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"c56383f08ecbaa6bfca9d0b9ee0afbb34770d822","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 16:59:15.000000000","message":"Patch Set 39:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/39/check/DellEMC_XtremIO_FC/01f57a6/DellEMC_XtremIO_FC/91 : SUCCESS in 53m 19s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/39/check/DellEMC_XtremIO_iSCSI/3b4ee43/DellEMC_XtremIO_iSCSI/89 : SUCCESS in 1h 19m 37s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":39},{"id":"3d97071711c94bda4d56584caebef757121302f5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-30 17:04:04.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/39/upstream-check/cinder-cDOT-iSCSI/766a00b/ : FAILURE in 1h 23m 55s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/39/upstream-check/cinder-cDOT-NFS/baf055d/ : FAILURE in 1h 15m 45s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/39/upstream-check/cinder-cDOT-FCP/1ac661b/ : FAILURE in 1h 07m 52s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":39},{"id":"90990598e10a7dccd87710f548fa6a7bffe164e4","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-04-30 17:14:35.000000000","message":"Patch Set 39:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/65/621465/39/check/DellEMC_SC_FC/b821295/DellEMC_SC_FC/15 : SUCCESS in 1h 05m 14s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/65/621465/39/check/DellEMC_SC_iSCSI/7d11876/DellEMC_SC_iSCSI/14 : FAILURE in 1h 34m 59s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":39},{"id":"dd11a7c7ca506c64a1975d1b27661d662ebedf32","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-30 17:15:00.000000000","message":"Patch Set 39:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/39 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"ae043618fe69818666c209bf521d185856061f04","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 17:15:46.000000000","message":"Patch Set 39:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/65/621465/39/check/DellEMC_VNX_iSCSI/ebd5fd0/DellEMC_VNX_iSCSI/26 : SUCCESS in 1h 36m 11s\n- DellEMC_VNX_FC http://publiclogs.emc.com/65/621465/39/check/DellEMC_VNX_FC/ce073d7/DellEMC_VNX_FC/27 : FAILURE in 56m 37s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":39},{"id":"19a48cc62a8d7fe15550b7b68eb95ce0fe642934","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-04-30 17:39:01.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/39/upstream-solidfire-check/cinder-solidfire-iSCSI/7c562b8/ : SUCCESS in 1h 58m 22s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":39},{"id":"761a4d118f2b80f5948966d333ab6c83d3647b40","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-04-30 17:52:23.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/39/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/420fcba : SUCCESS in 2h 08m 13s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/39/check/3par-fc-driver-master-client-pip-ssa02-dsvm/bbf97c8 : SUCCESS in 1h 55m 57s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/39/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/e8ea86a : SUCCESS in 2h 10m 28s","accounts_in_message":[],"_revision_number":39},{"id":"714ee93092e421fa88ec77ad110998a0268dd45a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 18:40:20.000000000","message":"Patch Set 39:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/65/621465/39/check/DellEMC_PowerMAX_FC/57f66b5/DellEMC_PowerMAX_FC/23 : SUCCESS in 2h 23m 34s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/65/621465/39/check/DellEMC_PowerMAX_iSCSI/bd816a0/DellEMC_PowerMAX_iSCSI/24 : SUCCESS in 3h 00m 49s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":39},{"id":"a094e70f83c4fc27d3ebee7ea51e497818ef2949","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-30 20:05:56.000000000","message":"Patch Set 39:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/39 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"9e7e3c2bf4f8c6622fb2b5a77fe3a954fb2b30f0","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-04-30 20:11:04.000000000","message":"Patch Set 39: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":39},{"id":"4c1744691dbf4dfe888e1fcef56d8dcd1f272fef","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-04-30 21:25:14.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"8466831b572b2f1684dbcd66caa95a7bd4ce95cd","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-04-30 21:25:31.000000000","message":"Patch Set 40:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":40},{"id":"2c69e2036fdf29aaadfe12366e12498d1a44b5ee","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-04-30 21:25:45.000000000","message":"Patch Set 40:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/40 : FAILURE","accounts_in_message":[],"_revision_number":40},{"id":"cbb7b3f06ef58a49537620a9c7ee6e0536951028","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 21:26:36.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/40/check/dsvm-tempest-kvm/17d182f : FAILURE in 3s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/40/check/dsvm-tempest-exe/afcdac3 : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":40},{"id":"c9ba5a34b681765ce0893839b6d444c50ab5f1f4","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-04-30 21:30:12.000000000","message":"Patch Set 40:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/40/check/dsvm-tempest-zadara-driver/c41d0e5 : FAILURE in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"62c9d39ac0df01b5ba802fac8adbe7b7a3a69b39","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-30 21:30:22.000000000","message":"Patch Set 40:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/40/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f42ef4f : FAILURE in 4m 42s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/40/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/8a58b89 : FAILURE in 4m 09s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":40},{"id":"0b3433c89158eaf3b3001c69e9db3fc062ce5a7d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-30 22:48:35.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/40/check/dsvm-tempest-infinibox-fc/ff018fa : SUCCESS in 54m 30s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/40/check/dsvm-tempest-infinibox-iscsi/b7764de : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":40},{"id":"10ba454b255e8ec8df578ef48544f452ed4e6eab","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 22:51:56.000000000","message":"Patch Set 40:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/40/check/DellEMC_XtremIO_FC/ae74c22/DellEMC_XtremIO_FC/93 : SUCCESS in 55m 02s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/40/check/DellEMC_XtremIO_iSCSI/ad1e9e9/DellEMC_XtremIO_iSCSI/91 : SUCCESS in 1h 26m 11s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":40},{"id":"0c25d76bbb60c9377a3c8523f45c25a86d79cabe","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-04-30 22:56:29.000000000","message":"Patch Set 40:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"1d0a89ba55a564025873b3cf5ad5f734b1d4d5df","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-04-30 22:56:41.000000000","message":"Patch Set 40:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/40/check/fujitsu-eternus-dx-iscsi/dc53d32 : FAILURE in 12m 11s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/40/check/fujitsu-eternus-dx-fc/0510c86 : SUCCESS in 1h 30m 41s","accounts_in_message":[],"_revision_number":40},{"id":"06262cb04ab2e3a7aaa8d6c95b4352ddd9281fb8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-04-30 22:57:59.000000000","message":"Patch Set 40:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/40/check/DellEMC_Unity_FC/8867d66/DellEMC_Unity_FC/69 : SUCCESS in 1h 18m 01s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/40/check/DellEMC_Unity_iSCSI/c86f75f/DellEMC_Unity_iSCSI/69 : SUCCESS in 1h 32m 25s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":40},{"id":"079fc6f5947def12fb5c74bf940dbb8e6db58352","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 23:26:32.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/40/check/NEC-ISCSI/672ebe2 : SUCCESS in 2h 00m 36s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/40/check/NEC-FC/73a8e8f : SUCCESS in 2h 00m 45s","accounts_in_message":[],"_revision_number":40},{"id":"feb6d28c7bbc5b579bf0847561a65beae7de669e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-04-30 23:31:06.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/40/check/dsvm-tempest-storpool/f6e8c5e : SUCCESS in 2h 05m 16s","accounts_in_message":[],"_revision_number":40},{"id":"99392d3bc0f4754aa981a9d7791b114971768716","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 00:01:15.000000000","message":"Patch Set 40:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/65/621465/40/check/DellEMC_PowerMAX_FC/627db5d/DellEMC_PowerMAX_FC/25 : FAILURE in 2h 35m 41s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/65/621465/40/check/DellEMC_PowerMAX_iSCSI/7ae7119/DellEMC_PowerMAX_iSCSI/26 : SUCCESS in 2h 34m 01s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":40},{"id":"2dcec9955454cb90d3018b2e5881133638d5b1be","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-01 00:29:35.000000000","message":"Patch Set 40:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"c6b34df7cfbfd2dbc915f820041388a8ff533ab5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 00:46:17.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/40/check/openstack-tox-lower-constraints/7d43c90/ : SUCCESS in 10m 39s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/40/check/openstack-tox-pep8/258152a/ : SUCCESS in 7m 46s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/40/check/openstack-tox-py27/1dd4a14/ : SUCCESS in 10m 14s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/40/check/openstack-tox-py36/8cfbd9f/ : SUCCESS in 10m 21s\n- openstack-tox-docs http://logs.openstack.org/65/621465/40/check/openstack-tox-docs/ae571bf/html/ : SUCCESS in 9m 17s\n- requirements-check http://logs.openstack.org/65/621465/40/check/requirements-check/36f394d/ : SUCCESS in 4m 45s\n- tempest-full http://logs.openstack.org/65/621465/40/check/tempest-full/82de995/ : SUCCESS in 1h 54m 28s\n- neutron-grenade http://logs.openstack.org/65/621465/40/check/neutron-grenade/a416e4d/ : SUCCESS in 1h 11m 05s\n- grenade-py3 http://logs.openstack.org/65/621465/40/check/grenade-py3/a146680/ : SUCCESS in 1h 11m 03s\n- tempest-full-py3 http://logs.openstack.org/65/621465/40/check/tempest-full-py3/32a3f68/ : SUCCESS in 1h 34m 43s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/40/check/build-openstack-releasenotes/92045ac/html/ : SUCCESS in 13m 19s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/40/check/cinder-tempest-dsvm-lvm-lio-barbican/582d9ef/ : SUCCESS in 1h 44m 45s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/40/check/cinder-tox-bandit-baseline/7e96db8/ : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/40/check/openstack-tox-functional-py36/0e6e49a/ : SUCCESS in 5m 24s\n- cinder-rally-task http://logs.openstack.org/65/621465/40/check/cinder-rally-task/51ddbe4/ : SUCCESS in 40m 16s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/40/check/openstack-tox-pylint/07e951f/ : SUCCESS in 4m 53s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/40/check/devstack-plugin-ceph-tempest/825131d/ : SUCCESS in 1h 24m 58s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/40/check/openstack-tox-functional/1213ad2/ : SUCCESS in 5m 49s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/40/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/68777a6/ : SUCCESS in 1h 10m 06s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/40/check/legacy-tempest-dsvm-lvm-multibackend/600547d/ : SUCCESS in 1h 27m 39s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/40/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/923a65f/ : SUCCESS in 1h 29m 45s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/40/check/tempest-slow-py3/f1f749a/ : SUCCESS in 2h 17m 57s","accounts_in_message":[],"_revision_number":40},{"id":"c193fa4fb5aaf33cb89e6dfae0cd519f89ebdf16","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-05-01 01:00:26.000000000","message":"Patch Set 40:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-40/nedge_iscsi : FAILURE in 11m 25s","accounts_in_message":[],"_revision_number":40},{"id":"071af9c1510367576b60b386e3a2a19693c25d6f","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-05-01 01:11:50.000000000","message":"Patch Set 40:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/65/621465/40/check/synology-dsm-cinder-iscsi/d6adbf4 : FAILURE in 25m 03s","accounts_in_message":[],"_revision_number":40},{"id":"69ffd48d04ba624e8e6c308b7734673e8fccac5c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-05-01 01:12:39.000000000","message":"Patch Set 40:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-40/ns5_iscsi : FAILURE in 11m 13s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-40/ns5_nfs : FAILURE in 23m 38s","accounts_in_message":[],"_revision_number":40},{"id":"93eefe7e3c26129b350b334614ee6de1e8bf49a9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 01:25:46.000000000","message":"Patch Set 40:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/65/621465/40/check/DellEMC_VxFlexOS/577131f/DellEMC_VxFlexOS/15 : FAILURE in 28s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":40},{"id":"677948d210eab3690c1451eed13b0f8866f0d633","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-05-01 01:56:09.000000000","message":"Patch Set 40:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-40 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI","accounts_in_message":[],"_revision_number":40},{"id":"ba3b0c9d83460519d97f3d02c8b8e6ca46c880c7","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-05-01 01:59:06.000000000","message":"Patch Set 40:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/1877 : FAILURE","accounts_in_message":[],"_revision_number":40},{"id":"ce21db8d8d4060796dc74d227de81eafacf679f4","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 02:05:08.000000000","message":"Patch Set 40:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-40-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"4a341b4aed18a048efffcdfada7e04f950b9ccf4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-05-01 02:08:37.000000000","message":"Patch Set 40:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-40 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":40},{"id":"0506ae6b9068dd70839b4a0e21db42a235262e3d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 02:09:40.000000000","message":"Patch Set 40:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/40/check/DellEMC_XtremIO_FC/5dd4141/DellEMC_XtremIO_FC/94 : SUCCESS in 53m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/40/check/DellEMC_XtremIO_iSCSI/5bbe6d8/DellEMC_XtremIO_iSCSI/92 : SUCCESS in 1h 23m 04s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":40},{"id":"6664f520bb8a100ed618f038e089aaf2e0310b4f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-05-01 02:10:52.000000000","message":"Patch Set 40:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/65/621465/40/check/DellEMC_SC_FC/31d522b/DellEMC_SC_FC/17 : FAILURE in 1h 06m 39s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/65/621465/40/check/DellEMC_SC_iSCSI/6baddb7/DellEMC_SC_iSCSI/16 : FAILURE in 1h 24m 15s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":40},{"id":"1a9249e5099653993f096590bd590f95aa8c2bd4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 02:22:02.000000000","message":"Patch Set 40:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/65/621465/40/check/DellEMC_VNX_iSCSI/bf99966/DellEMC_VNX_iSCSI/28 : SUCCESS in 1h 35m 26s\n- DellEMC_VNX_FC http://publiclogs.emc.com/65/621465/40/check/DellEMC_VNX_FC/4bcb078/DellEMC_VNX_FC/29 : SUCCESS in 54m 52s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":40},{"id":"588b1243646d7393df91e36e37db2ed40a6577f0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-05-01 02:24:03.000000000","message":"Patch Set 40:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/40/check/DellEMC_Unity_FC/baee95d/DellEMC_Unity_FC/70 : SUCCESS in 1h 02m 46s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/40/check/DellEMC_Unity_iSCSI/9a00561/DellEMC_Unity_iSCSI/70 : SUCCESS in 1h 37m 35s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":40},{"id":"eb3423a37f556318e6e9511d8012f11968203312","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-01 02:45:52.000000000","message":"Patch Set 40:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"225e5acb02eb677de00f7ac445b205dafac06771","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-05-01 03:05:58.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/40/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/726504b : SUCCESS in 2h 17m 03s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/40/check/3par-fc-driver-master-client-pip-ssa02-dsvm/f27766f : SUCCESS in 1h 59m 57s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/40/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/2f632b8 : SUCCESS in 2h 02m 21s","accounts_in_message":[],"_revision_number":40},{"id":"8956335f090da3e672a8a1caf1b642959aadf02d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 03:22:20.000000000","message":"Patch Set 40:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/65/621465/40/check/DellEMC_PowerMAX_FC/7700c56/DellEMC_PowerMAX_FC/26 : SUCCESS in 2h 22m 47s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/65/621465/40/check/DellEMC_PowerMAX_iSCSI/c1df6d7/DellEMC_PowerMAX_iSCSI/27 : SUCCESS in 2h 35m 49s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":40},{"id":"245cf87d485db54b567f4ab698ef7cb73e220f80","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-01 04:54:45.000000000","message":"Patch Set 40:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"96849db698da9931ac5441c3d14d03d80f002818","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-01 07:38:40.000000000","message":"Patch Set 40:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/40/iscsi/ : SUCCESS in 57m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/40/smb/ : FAILURE in 40m 19s","accounts_in_message":[],"_revision_number":40},{"id":"48eff13ac5532e407dcff455425e80a4571ba597","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-01 09:13:13.000000000","message":"Patch Set 40:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"a3047ef846ee9d9ea4a1efede0cddb5c70c136bd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-01 11:27:55.000000000","message":"Patch Set 40:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/40/check-not-vote/ext-cinder-zuul/0e117db : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":40},{"id":"70c807de379bd90c87da943e57fe785c853858fc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-01 18:32:57.000000000","message":"Patch Set 40:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/40/check/tempest-dsvm-full-xenial/9ee4f2b/ : SUCCESS in 2h 43m 10s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/40/check/devstack-plugin-ceph-dsvm-xenial/51cdb4d/ : SUCCESS in 1h 31m 04s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"41a0189a5517e3eef6177012dc96316d82d1509c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-01 20:06:10.000000000","message":"Patch Set 40:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"d01b664185da61279662a10eb4211f28d2526473","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-01 22:12:17.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/40/upstream-check/cinder-cDOT-iSCSI/aeb2e7a/ : FAILURE in 1h 59m 44s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/40/upstream-check/cinder-cDOT-NFS/12a3a28/ : FAILURE in 1h 12m 37s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/40/upstream-check/cinder-cDOT-FCP/758b780/ : SUCCESS in 58m 09s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":40},{"id":"ba73cb232fc3d35f7aa6bf5729c0f76a4e4b3d96","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-05-01 23:34:21.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/40/upstream-solidfire-check/cinder-solidfire-iSCSI/554be6e/ : FAILURE in 3h 21m 22s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":40},{"id":"4550d9714566310474e4aa4c5e0f70328889eb3f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-01 23:44:41.000000000","message":"Patch Set 40:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"4fbf355c9abf854188bf6c38b854a90bf5b59847","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-05-02 21:07:48.000000000","message":"Patch Set 40: Review-Priority+1","accounts_in_message":[],"_revision_number":40},{"id":"8ae43cfa6acb05b3bdd2a05ff169c5425d3ec9be","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-05-02 21:14:16.000000000","message":"Patch Set 40: Code-Review-1\n\nPlease enable the cinder_tempest_plugin tempest tests in the RSD CI.","accounts_in_message":[],"_revision_number":40},{"id":"c940f904383ed85209c238692a4241710bba349f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-05-02 21:16:37.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"09d85da8d2720b460995de1cd94aa2f5928aabe5","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-02 21:17:02.000000000","message":"Patch Set 40: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":40},{"id":"07fb0403bc97dc014f145227dd733d010594c4ae","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-02 21:19:41.000000000","message":"Patch Set 40: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":40},{"id":"25f78f13c18d32d3e5582e0e4be22a14491ce886","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-05-02 21:20:19.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"2befe68866dfa4e33ff7879b1428bf6de2b87023","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-03 07:33:33.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"d179351c280ac364f83b73162b2bafece3cf5b12","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-05-03 07:33:48.000000000","message":"Patch Set 41:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":41},{"id":"9a23a7442c83d2ca51c4ab32d498178a0cfa4447","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-05-03 07:34:09.000000000","message":"Patch Set 41:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/41 : FAILURE","accounts_in_message":[],"_revision_number":41},{"id":"4949876703fac757f866c96deadc2c5e04da4d81","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-03 07:34:58.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/41/check/dsvm-tempest-kvm/5bfdc14 : FAILURE in 4s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/41/check/dsvm-tempest-exe/91bfa94 : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":41},{"id":"7efc7fa507c19d4a021744dae914893e749cf68d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-03 07:39:12.000000000","message":"Patch Set 41:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/41/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/e130d17 : FAILURE in 4m 27s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/41/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/0d6282b : FAILURE in 5m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":41},{"id":"282865e393a1161fc792319880aa957052e17a91","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-03 07:47:14.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/41/check/dsvm-tempest-zadara-driver/4c879c0 : FAILURE in 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"d642035822c5b6975ba359562c5fd6b8bdc52b4b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-03 08:10:57.000000000","message":"Patch Set 41:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/41/check-not-vote/ext-cinder-zuul/dc7ee71 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":41},{"id":"cc3ed266c910473a52854df90e56b53a520aded1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-03 08:26:21.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/41/iscsi/ : SUCCESS in 52m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/41/smb/ : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":41},{"id":"3a704c09389da3f56e5b689034326e4bc303f173","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-05-03 08:46:20.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/41/check/dsvm-tempest-infinibox-fc/aeb0159 : SUCCESS in 1h 07m 19s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/41/check/dsvm-tempest-infinibox-iscsi/02af4c3 : SUCCESS in 1h 12m 18s","accounts_in_message":[],"_revision_number":41},{"id":"b2e617468380b836387050ff66d8aad8cb01dd83","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-05-03 08:50:54.000000000","message":"Patch Set 41:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/41/check/DellEMC_Unity_FC/7d5edaa/DellEMC_Unity_FC/98 : SUCCESS in 1h 14m 52s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/41/check/DellEMC_Unity_iSCSI/25a9a19/DellEMC_Unity_iSCSI/97 : SUCCESS in 1h 17m 06s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":41},{"id":"eaeac99e7516df26b25e31a146343b7fe588be30","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-05-03 08:52:10.000000000","message":"Patch Set 41:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/41/check-cinder/Cinder-ISER-ISCSI/cf4be09 : FAILURE in 1h 17m 31s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/41/check-cinder/Cinder-ISER-LIO/5eae834 : SUCCESS in 1h 09m 04s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":41},{"id":"1c16eea812611d655c63f1039ecd7b4f73cfa58c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-03 08:53:08.000000000","message":"Patch Set 41:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/41/check/DellEMC_XtremIO_FC/27df4fc/DellEMC_XtremIO_FC/126 : SUCCESS in 50m 05s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/41/check/DellEMC_XtremIO_iSCSI/fb1f79a/DellEMC_XtremIO_iSCSI/120 : SUCCESS in 1h 19m 12s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":41},{"id":"8a916087c4ad5aa1c08ed67641dbfdd4d77971fb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-05-03 08:58:44.000000000","message":"Patch Set 41:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/41/check/fujitsu-eternus-dx-iscsi/0c29c0d : FAILURE in 10m 23s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/41/check/fujitsu-eternus-dx-fc/3c44283 : FAILURE in 1h 24m 28s","accounts_in_message":[],"_revision_number":41},{"id":"e684fd2f19a20a6c8b86a8ce29f49b48b24ee32f","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-05-03 09:00:51.000000000","message":"Patch Set 41:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-41 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI","accounts_in_message":[],"_revision_number":41},{"id":"88a588d33aa8cc01861117ff17041e6388971bb3","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-03 09:05:30.000000000","message":"Patch Set 41:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/41 : SUCCESS","accounts_in_message":[],"_revision_number":41},{"id":"3676cdc2f6acdb76794696ba401c14a1dda9ea3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 09:56:58.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/41/check/openstack-tox-lower-constraints/a04c282/ : SUCCESS in 11m 29s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/41/check/openstack-tox-pep8/3f60d1e/ : FAILURE in 7m 03s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/41/check/openstack-tox-py27/e846ae8/ : SUCCESS in 8m 02s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/41/check/openstack-tox-py36/ed376d1/ : SUCCESS in 7m 29s\n- openstack-tox-docs http://logs.openstack.org/65/621465/41/check/openstack-tox-docs/2f395f7/html/ : SUCCESS in 7m 35s\n- requirements-check http://logs.openstack.org/65/621465/41/check/requirements-check/0dff205/ : SUCCESS in 2m 31s\n- tempest-full http://logs.openstack.org/65/621465/41/check/tempest-full/fcb84c6/ : SUCCESS in 1h 53m 54s\n- neutron-grenade http://logs.openstack.org/65/621465/41/check/neutron-grenade/4b9f066/ : SUCCESS in 1h 02m 17s\n- grenade-py3 http://logs.openstack.org/65/621465/41/check/grenade-py3/d06b932/ : SUCCESS in 1h 13m 50s\n- tempest-full-py3 http://logs.openstack.org/65/621465/41/check/tempest-full-py3/15f3898/ : SUCCESS in 1h 33m 34s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/41/check/build-openstack-releasenotes/94a0ce3/html/ : SUCCESS in 14m 17s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/41/check/cinder-tempest-dsvm-lvm-lio-barbican/9ebf354/ : SUCCESS in 1h 27m 57s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/41/check/cinder-tox-bandit-baseline/4ef684b/ : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/41/check/openstack-tox-functional-py36/478aa6b/ : SUCCESS in 4m 40s\n- cinder-rally-task http://logs.openstack.org/65/621465/41/check/cinder-rally-task/5f50d19/ : SUCCESS in 32m 33s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/41/check/openstack-tox-pylint/29f1f7b/ : SUCCESS in 4m 16s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/41/check/devstack-plugin-ceph-tempest/d20f36a/ : SUCCESS in 1h 23m 02s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/41/check/openstack-tox-functional/cbdf7cb/ : SUCCESS in 4m 36s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/41/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/83c3e78/ : SUCCESS in 56m 18s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/41/check/legacy-tempest-dsvm-lvm-multibackend/a6b5479/ : SUCCESS in 1h 35m 53s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/41/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2938375/ : SUCCESS in 1h 21m 48s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/41/check/tempest-slow-py3/b39a1ac/ : SUCCESS in 2h 17m 27s","accounts_in_message":[],"_revision_number":41},{"id":"d5f9d6614c0fb0949325276850dae51de215e9d5","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-05-03 10:29:34.000000000","message":"Patch Set 41:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-41 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI for rechecking and info.","accounts_in_message":[],"_revision_number":41},{"id":"d5bc5684e387d5bb8fa8d154245910a946be7920","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-03 10:40:06.000000000","message":"Patch Set 41:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/41 : SUCCESS","accounts_in_message":[],"_revision_number":41},{"id":"c737fdc832fe635f00cb7f74bd00aae66665995b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-03 11:32:20.000000000","message":"Patch Set 41:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/41/check/tempest-dsvm-full-xenial/c80d4ec/ : SUCCESS in 1h 15m 40s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/41/check/devstack-plugin-ceph-dsvm-xenial/966177e/ : SUCCESS in 1h 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"5b82a74b54ed40b30ec52ac0a3691ae512b60795","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-03 13:45:09.000000000","message":"Patch Set 41:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/41 : SUCCESS","accounts_in_message":[],"_revision_number":41},{"id":"257e18cafd586f1612ad2e600dca3ccdec09904a","author":{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},"date":"2019-05-03 15:48:58.000000000","message":"Patch Set 40:\n\n(5 comments)","accounts_in_message":[],"_revision_number":40},{"id":"2cd47e7c9c34bec7411e7c6183821e49c1165912","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-05-03 16:30:27.000000000","message":"Patch Set 41:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":41},{"id":"62a2c018483166afad85233a3625f51d64a63818","author":{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},"date":"2019-05-03 16:32:26.000000000","message":"Topic changed from train-drivers to train-driversel","accounts_in_message":[],"_revision_number":41},{"id":"1b5a8ee651c7821d9b6dc5f1b6e7ab29c0bccd8e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-05-03 17:21:27.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/41/check/NEC-ISCSI/8ccb03c : SUCCESS in 2h 35m 38s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/41/check/NEC-FC/ae1dda0 : SUCCESS in 2h 30m 56s","accounts_in_message":[],"_revision_number":41},{"id":"78bca3a17d0977a23cceb03aff1612e756071cc9","author":{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},"date":"2019-05-03 17:46:42.000000000","message":"Topic changed from train-driversel to train-drivers","accounts_in_message":[],"_revision_number":41},{"id":"193e6309dee3bfab6a05c3d28e774e99da1e1ccd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-05-03 17:51:25.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/41/check/dsvm-tempest-storpool/03ab3c3 : SUCCESS in 2h 04m 55s","accounts_in_message":[],"_revision_number":41},{"id":"e83436562ca282d1be620fa60a55b5a76ed3e6bd","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-03 18:04:03.000000000","message":"Patch Set 41:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/41 : SUCCESS","accounts_in_message":[],"_revision_number":41},{"id":"5e54ea3a3cb209c125557bc8865a684c07008ef3","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-03 18:34:44.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"f9e776916558a84331f4f2e6a1c5ed6b42d9376a","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-05-03 18:34:59.000000000","message":"Patch Set 42:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":42},{"id":"eb7034aa314905e6fe8e03343a51a9e465716c79","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-05-03 18:35:15.000000000","message":"Patch Set 42:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/42 : FAILURE","accounts_in_message":[],"_revision_number":42},{"id":"1055a99c0231bc9b48a5c2943837c64f5773f8ba","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-03 18:36:08.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/42/check/dsvm-tempest-kvm/0529fae : FAILURE in 2s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/42/check/dsvm-tempest-exe/a10aa4e : FAILURE in 4s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":42},{"id":"58484c691beb64d67fe4e69ea988a681f57ec84c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-03 18:39:35.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/42/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/2235373 : FAILURE in 4m 01s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/42/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/d654da2 : FAILURE in 4m 28s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":42},{"id":"21eb5cef35edf22e30bd671a3d754b07eb0b584a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-03 18:48:54.000000000","message":"Patch Set 42:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/42/check/dsvm-tempest-zadara-driver/cff46c8 : FAILURE in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"9e3c58269ab1b8e873e74892049b32f354636f32","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-03 19:30:08.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/42/iscsi/ : SUCCESS in 54m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/42/smb/ : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":42},{"id":"8c4d47673e2e9e882327aa54f6ac06351976a2c7","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-03 19:50:45.000000000","message":"Patch Set 42:\n\n(13 comments)\n\n\u003e Please enable the cinder_tempest_plugin tempest tests in the RSD\n \u003e CI.\n\ninstalled tempest plugin and completed RSD CI successfully. tempest will automatically discover the installed plugin. Please let me know whether this is the right approach.","accounts_in_message":[],"_revision_number":42},{"id":"18fe988a90b3587f5c6ae48eb853404cc773aeec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-03 19:59:10.000000000","message":"Patch Set 42:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/42/check/tempest-dsvm-full-xenial/f0f6f2d/ : SUCCESS in 1h 22m 57s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/42/check/devstack-plugin-ceph-dsvm-xenial/53e3919/ : SUCCESS in 1h 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"1ea869b5378181851d9761af35112e94aba0ba82","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-05-03 20:03:58.000000000","message":"Patch Set 42:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/42/check/fujitsu-eternus-dx-iscsi/a7f7d81 : FAILURE in 10m 56s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/42/check/fujitsu-eternus-dx-fc/916c866 : SUCCESS in 1h 28m 30s","accounts_in_message":[],"_revision_number":42},{"id":"dbc979770239b270ef941c8bf0976ebe3f70ffec","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-03 20:07:56.000000000","message":"Patch Set 42:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"5ad896c459508daaeba70c378ba816c46b73fbea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:23:01.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/42/check/openstack-tox-lower-constraints/645e2fa/ : SUCCESS in 10m 28s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/42/check/openstack-tox-pep8/4cdff9f/ : SUCCESS in 7m 27s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/42/check/openstack-tox-py27/a60542b/ : SUCCESS in 8m 30s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/42/check/openstack-tox-py36/1fed72e/ : SUCCESS in 9m 24s\n- openstack-tox-docs http://logs.openstack.org/65/621465/42/check/openstack-tox-docs/68b7ef7/html/ : SUCCESS in 7m 43s\n- requirements-check http://logs.openstack.org/65/621465/42/check/requirements-check/6693db5/ : SUCCESS in 2m 53s\n- tempest-full http://logs.openstack.org/65/621465/42/check/tempest-full/d26776b/ : SUCCESS in 1h 36m 00s\n- neutron-grenade http://logs.openstack.org/65/621465/42/check/neutron-grenade/3619ac3/ : SUCCESS in 1h 01m 04s\n- grenade-py3 http://logs.openstack.org/65/621465/42/check/grenade-py3/0a55d24/ : SUCCESS in 1h 10m 37s\n- tempest-full-py3 http://logs.openstack.org/65/621465/42/check/tempest-full-py3/21691c1/ : SUCCESS in 1h 32m 27s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/42/check/build-openstack-releasenotes/f4f2824/html/ : SUCCESS in 16m 37s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/42/check/cinder-tempest-dsvm-lvm-lio-barbican/e0698f9/ : FAILURE in 46m 15s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/42/check/cinder-tox-bandit-baseline/5267e33/ : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/42/check/openstack-tox-functional-py36/bb94587/ : SUCCESS in 6m 46s\n- cinder-rally-task http://logs.openstack.org/65/621465/42/check/cinder-rally-task/b54af6f/ : SUCCESS in 43m 25s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/42/check/openstack-tox-pylint/a0f1a21/ : SUCCESS in 5m 06s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/42/check/devstack-plugin-ceph-tempest/087b2fe/ : SUCCESS in 1h 38m 00s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/42/check/openstack-tox-functional/3de40d6/ : SUCCESS in 5m 02s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/42/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/ce5babc/ : FAILURE in 55m 19s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/42/check/legacy-tempest-dsvm-lvm-multibackend/626a7cb/ : SUCCESS in 1h 33m 15s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/42/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/dcc3ea3/ : SUCCESS in 1h 34m 00s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/42/check/tempest-slow-py3/c9ee0fe/ : SUCCESS in 2h 24m 23s","accounts_in_message":[],"_revision_number":42},{"id":"a201ec2cc5942d1c3e68a2fef97a192a58baeb39","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-05-03 21:26:02.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/42/check/dsvm-tempest-infinibox-fc/3d9599c : SUCCESS in 54m 43s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/42/check/dsvm-tempest-infinibox-iscsi/fa23e89 : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":42},{"id":"da73a910ea6be42e21673fbd1190228b3f97518d","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-03 21:27:28.000000000","message":"Patch Set 42:\n\nrecheck","accounts_in_message":[],"_revision_number":42},{"id":"f3ec9e557d0e2790b157bae08022518336086590","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-05-03 21:27:43.000000000","message":"Patch Set 42:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":42},{"id":"667377cc17db633d37ea92f55a5ac9f4ffc04ab4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-03 21:28:47.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/42/check/dsvm-tempest-kvm/72c4c9a : FAILURE in 2s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/42/check/dsvm-tempest-exe/ae58f49 : FAILURE in 4s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":42},{"id":"3cae6cab66588b221e2b7b1b40d346b2242b76ec","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-03 21:32:21.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/42/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/45697fe : FAILURE in 3m 36s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/42/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/3416b45 : FAILURE in 4m 35s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":42},{"id":"0b0feeb9850d72a2f874149239f543229aefc7d1","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-03 22:48:57.000000000","message":"Patch Set 42:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"3450f624628e38f3d55ab8e95306d79f48aed541","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-03 22:54:49.000000000","message":"Patch Set 42:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/42/iscsi/ : FAILURE in 1h 26m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/42/smb/ : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":42},{"id":"8dfc603681b4829a80be03120a9bc8c2965ffb39","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-03 22:55:44.000000000","message":"Patch Set 42:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/42/check/tempest-dsvm-full-xenial/fbcf8ba/ : SUCCESS in 1h 27m 48s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/42/check/devstack-plugin-ceph-dsvm-xenial/47731d3/ : FAILURE in 1h 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"6c1bbcc2aee63acb22c71379962f77bfb0d1c388","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-03 23:20:10.000000000","message":"Patch Set 42:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/42/check/dsvm-tempest-zadara-driver/216ea47 : FAILURE in 1h 52m 28s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"44ba79cc1ce07b04602b932446e19cfcc3c107e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-03 23:56:22.000000000","message":"Patch Set 42:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/42/check-not-vote/ext-cinder-zuul/1b69b3a : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":42},{"id":"76c26d28b733021f35ef4faef8ba5964b89571bc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 00:07:54.000000000","message":"Patch Set 42:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/42/check/DellEMC_XtremIO_FC/d79899b/DellEMC_XtremIO_FC/135 : SUCCESS in 51m 20s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/42/check/DellEMC_XtremIO_iSCSI/5b03bf6/DellEMC_XtremIO_iSCSI/129 : SUCCESS in 1h 20m 02s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":42},{"id":"2b5b5fa388e4b67988aa88a94a1b0f31ec2d823f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-05-04 00:13:17.000000000","message":"Patch Set 42:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/42/check/DellEMC_Unity_FC/2c94f11/DellEMC_Unity_FC/107 : SUCCESS in 1h 02m 38s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/42/check/DellEMC_Unity_iSCSI/2601eb6/DellEMC_Unity_iSCSI/106 : SUCCESS in 1h 14m 55s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":42},{"id":"984a1928918ba7e852f70e836b5d246670254b81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 00:21:49.000000000","message":"Patch Set 42:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/42/check/openstack-tox-lower-constraints/5145e7e/ : SUCCESS in 9m 46s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/42/check/openstack-tox-pep8/67f6903/ : SUCCESS in 7m 25s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/42/check/openstack-tox-py27/585088d/ : FAILURE in 9m 31s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/42/check/openstack-tox-py36/428f55b/ : SUCCESS in 8m 10s\n- openstack-tox-docs http://logs.openstack.org/65/621465/42/check/openstack-tox-docs/4c693ad/html/ : SUCCESS in 7m 56s\n- requirements-check http://logs.openstack.org/65/621465/42/check/requirements-check/b2478d7/ : SUCCESS in 2m 45s\n- tempest-full http://logs.openstack.org/65/621465/42/check/tempest-full/83b3a25/ : SUCCESS in 1h 50m 22s\n- neutron-grenade http://logs.openstack.org/65/621465/42/check/neutron-grenade/c29addd/ : SUCCESS in 59m 51s\n- grenade-py3 http://logs.openstack.org/65/621465/42/check/grenade-py3/861a176/ : SUCCESS in 1h 06m 53s\n- tempest-full-py3 http://logs.openstack.org/65/621465/42/check/tempest-full-py3/4b8a015/ : SUCCESS in 1h 29m 54s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/42/check/build-openstack-releasenotes/c5dc7d3/html/ : SUCCESS in 15m 43s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/42/check/cinder-tempest-dsvm-lvm-lio-barbican/1364002/ : SUCCESS in 1h 31m 33s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/42/check/cinder-tox-bandit-baseline/cd3dced/ : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/42/check/openstack-tox-functional-py36/39229ac/ : SUCCESS in 4m 56s\n- cinder-rally-task http://logs.openstack.org/65/621465/42/check/cinder-rally-task/e32aaeb/ : SUCCESS in 38m 35s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/42/check/openstack-tox-pylint/8dc4caa/ : SUCCESS in 5m 34s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/42/check/devstack-plugin-ceph-tempest/c1dbcd8/ : SUCCESS in 1h 29m 19s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/42/check/openstack-tox-functional/b9742bd/ : SUCCESS in 5m 35s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/42/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/7cc0764/ : SUCCESS in 1h 12m 14s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/42/check/legacy-tempest-dsvm-lvm-multibackend/994ff10/ : SUCCESS in 1h 37m 56s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/42/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/cad19dc/ : SUCCESS in 1h 37m 28s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/42/check/tempest-slow-py3/4244767/ : SUCCESS in 2h 04m 55s","accounts_in_message":[],"_revision_number":42},{"id":"b55f40c4bef99220f2183db44eff0ee87c4ab1b1","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-04 02:25:00.000000000","message":"Patch Set 42:\n\nrecheck","accounts_in_message":[],"_revision_number":42},{"id":"b1683d720b4356bcc2e16f01c45485aeed54847d","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-05-04 02:25:14.000000000","message":"Patch Set 42:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":42},{"id":"68d02e512df6d4e02ecdbffe694e70f8951d5556","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 02:26:17.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/42/check/dsvm-tempest-kvm/e8b426d : FAILURE in 3s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/42/check/dsvm-tempest-exe/f4c826c : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":42},{"id":"7d62896506a091d1b35f9fa9486063ecf29f06ff","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-04 02:30:18.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/42/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/b37ffb9 : FAILURE in 4m 52s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/42/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/d1c7972 : FAILURE in 4m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":42},{"id":"31c8a595ef3392e11921b9cf1ad38058c53b9bc8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-04 02:39:04.000000000","message":"Patch Set 42:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/42/check/dsvm-tempest-zadara-driver/fd88929 : FAILURE in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"c5458f7928eeeb4325c8231955c9d68c5f2f34b2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-04 03:04:36.000000000","message":"Patch Set 42:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/42/check-not-vote/ext-cinder-zuul/78e748e : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":42},{"id":"02a6115d7f1d99e34c2479cbbea4dac9b3a6c58e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-04 03:12:44.000000000","message":"Patch Set 42:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"f88c4fdf4bbdfa4b7d22d377d2fd9b56d9f44616","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-04 03:33:16.000000000","message":"Patch Set 42:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/42/check/tempest-dsvm-full-xenial/99f51c0/ : SUCCESS in 1h 05m 05s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/42/check/devstack-plugin-ceph-dsvm-xenial/30e323b/ : SUCCESS in 1h 07m 59s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"72e4b415ff4909994e922342b43b2bce94131f4f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-04 03:52:58.000000000","message":"Patch Set 42:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/42/iscsi/ : FAILURE in 1h 27m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/42/smb/ : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":42},{"id":"59814fc2248a1996641df4e9955df9030df8f50a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 04:53:37.000000000","message":"Patch Set 42:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/42/check/openstack-tox-lower-constraints/5092be1/ : SUCCESS in 9m 27s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/42/check/openstack-tox-pep8/36a3d33/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/42/check/openstack-tox-py27/c31b7cc/ : SUCCESS in 7m 51s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/42/check/openstack-tox-py36/44f8e18/ : SUCCESS in 8m 25s\n- openstack-tox-docs http://logs.openstack.org/65/621465/42/check/openstack-tox-docs/5e5b7b1/html/ : SUCCESS in 6m 57s\n- requirements-check http://logs.openstack.org/65/621465/42/check/requirements-check/66f1cc0/ : SUCCESS in 3m 09s\n- tempest-full http://logs.openstack.org/65/621465/42/check/tempest-full/9c5a6d9/ : FAILURE in 1h 18m 40s\n- neutron-grenade http://logs.openstack.org/65/621465/42/check/neutron-grenade/94111a2/ : SUCCESS in 48m 39s\n- grenade-py3 http://logs.openstack.org/65/621465/42/check/grenade-py3/98d503c/ : SUCCESS in 48m 40s\n- tempest-full-py3 http://logs.openstack.org/65/621465/42/check/tempest-full-py3/0b6290e/ : SUCCESS in 1h 27m 47s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/42/check/build-openstack-releasenotes/34bc074/html/ : SUCCESS in 14m 07s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/42/check/cinder-tempest-dsvm-lvm-lio-barbican/8f1796e/ : SUCCESS in 1h 28m 55s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/42/check/cinder-tox-bandit-baseline/6262b37/ : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/42/check/openstack-tox-functional-py36/4a29e6d/ : SUCCESS in 4m 59s\n- cinder-rally-task http://logs.openstack.org/65/621465/42/check/cinder-rally-task/ac50883/ : SUCCESS in 36m 01s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/42/check/openstack-tox-pylint/2127ab2/ : SUCCESS in 4m 08s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/42/check/devstack-plugin-ceph-tempest/755bf7a/ : SUCCESS in 1h 27m 06s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/42/check/openstack-tox-functional/4f460ee/ : SUCCESS in 4m 55s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/42/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/d500c63/ : SUCCESS in 1h 02m 51s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/42/check/legacy-tempest-dsvm-lvm-multibackend/a9f0e28/ : SUCCESS in 1h 28m 47s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/42/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/e13bdeb/ : SUCCESS in 1h 40m 26s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/42/check/tempest-slow-py3/a9dd94c/ : SUCCESS in 2h 23m 12s","accounts_in_message":[],"_revision_number":42},{"id":"61112f1d2fe5e130e339aef503bfce4293ed7ec2","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-04 05:09:49.000000000","message":"Patch Set 42:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"e9a6e3403539030d20a986fcfd29fe85752242d3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 07:38:16.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/42/check/NEC-ISCSI/32b1c91 : SUCCESS in 2h 01m 46s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/42/check/NEC-FC/c96603d : SUCCESS in 2h 07m 02s","accounts_in_message":[],"_revision_number":42},{"id":"3e84e2d4ade3e57f8086aeceef04bd6c8a51835c","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-04 08:46:41.000000000","message":"Patch Set 42:\n\nrecheck","accounts_in_message":[],"_revision_number":42},{"id":"bb4e19c79aaf162e4efac7195bad89f4a65ca29e","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-05-04 08:46:56.000000000","message":"Patch Set 42:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":42},{"id":"7bc5476250daf3e90f93a0ae8dae7c1ca0659aef","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 08:48:03.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/42/check/dsvm-tempest-kvm/15bda72 : FAILURE in 4s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/42/check/dsvm-tempest-exe/3d7fc5f : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":42},{"id":"9c20c360ea8ef1edb152750e35745f3c55be31c9","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-04 08:52:02.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/42/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c1bb4d6 : FAILURE in 4m 02s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/42/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/8308981 : FAILURE in 5m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":42},{"id":"6cd7c4b44121a517d7e256f0b9ff6bc81d3a7960","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-04 09:00:30.000000000","message":"Patch Set 42:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/42/check/dsvm-tempest-zadara-driver/9639a1d : FAILURE in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"e1cd246373638caf11c9eb1a63a4bf01cf76d12c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-04 09:22:53.000000000","message":"Patch Set 42:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/42/check-not-vote/ext-cinder-zuul/7f696d6 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":42},{"id":"a8a94007775f1c28b21c36c1250381f6ceaf3f7a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-04 09:40:00.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/42/iscsi/ : SUCCESS in 52m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/42/smb/ : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":42},{"id":"0d611bac023684d875b3ee8e86c8038fcda9f74f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-04 10:14:55.000000000","message":"Patch Set 42:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/42/check/tempest-dsvm-full-xenial/29b1947/ : SUCCESS in 1h 23m 31s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/42/check/devstack-plugin-ceph-dsvm-xenial/8abcf3f/ : SUCCESS in 1h 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"1e879433c3150c1ad243714ff596331cbb07d603","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-04 10:19:58.000000000","message":"Patch Set 42:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"bb1f106444c0852aa1a0576b22e7e36c301c71df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 11:05:58.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/42/check/openstack-tox-lower-constraints/2bd2c08/ : SUCCESS in 9m 48s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/42/check/openstack-tox-pep8/848ed17/ : SUCCESS in 12m 22s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/42/check/openstack-tox-py27/0b7b933/ : SUCCESS in 12m 48s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/42/check/openstack-tox-py36/aabdf94/ : SUCCESS in 8m 54s\n- openstack-tox-docs http://logs.openstack.org/65/621465/42/check/openstack-tox-docs/966a147/html/ : SUCCESS in 12m 38s\n- requirements-check http://logs.openstack.org/65/621465/42/check/requirements-check/693a3c7/ : SUCCESS in 6m 47s\n- tempest-full http://logs.openstack.org/65/621465/42/check/tempest-full/5dc05cd/ : SUCCESS in 1h 40m 32s\n- neutron-grenade http://logs.openstack.org/65/621465/42/check/neutron-grenade/2d95815/ : SUCCESS in 46m 23s\n- grenade-py3 http://logs.openstack.org/65/621465/42/check/grenade-py3/c23f5ad/ : SUCCESS in 56m 56s\n- tempest-full-py3 http://logs.openstack.org/65/621465/42/check/tempest-full-py3/b785153/ : SUCCESS in 1h 18m 12s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/42/check/build-openstack-releasenotes/9ac87af/html/ : SUCCESS in 16m 22s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/42/check/cinder-tempest-dsvm-lvm-lio-barbican/3c3e293/ : SUCCESS in 1h 33m 11s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/42/check/cinder-tox-bandit-baseline/929c08d/ : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/42/check/openstack-tox-functional-py36/b03c4eb/ : SUCCESS in 8m 02s\n- cinder-rally-task http://logs.openstack.org/65/621465/42/check/cinder-rally-task/ed31cc1/ : SUCCESS in 40m 51s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/42/check/openstack-tox-pylint/f14a628/ : SUCCESS in 4m 21s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/42/check/devstack-plugin-ceph-tempest/28c8c96/ : SUCCESS in 1h 27m 23s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/42/check/openstack-tox-functional/a5b0311/ : SUCCESS in 4m 41s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/42/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/1e1de7a/ : SUCCESS in 59m 04s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/42/check/legacy-tempest-dsvm-lvm-multibackend/1049aed/ : SUCCESS in 1h 18m 45s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/42/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/461b413/ : SUCCESS in 1h 30m 55s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/42/check/tempest-slow-py3/989c7e7/ : SUCCESS in 1h 56m 33s","accounts_in_message":[],"_revision_number":42},{"id":"6e3230e698ae5bc73a1ab9d15fd9753c7f72a146","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 11:06:44.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/65/621465/42/check/DellEMC_VxFlexOS/1d0d29a/DellEMC_VxFlexOS/31 : FAILURE in 29s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":42},{"id":"6fc363c174cca95ac5954ec50908044cc2d71f13","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-05-04 11:10:37.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/42/check/dsvm-tempest-storpool/44b5072 : SUCCESS in 1h 56m 55s","accounts_in_message":[],"_revision_number":42},{"id":"bc18847697f4b7584306947713f6eecfc9145716","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-05-04 11:19:55.000000000","message":"Patch Set 42:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-42/nedge_iscsi : FAILURE in 11m 30s","accounts_in_message":[],"_revision_number":42},{"id":"55bc43cd5f5e4ebd24b1b60dd5ff65e8d7d752cb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-05-04 11:32:12.000000000","message":"Patch Set 42:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-42/ns5_iscsi : FAILURE in 11m 26s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-42/ns5_nfs : FAILURE in 23m 44s","accounts_in_message":[],"_revision_number":42},{"id":"be751269072024115beb3780d53264ebb8041779","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-05-04 11:32:47.000000000","message":"Patch Set 42:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/65/621465/42/check/synology-dsm-cinder-iscsi/790a7b6 : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":42},{"id":"ac1d3877abbc9ac3937d67447aa6a0ebaff90375","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-05-04 11:58:17.000000000","message":"Patch Set 42:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-42 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":42},{"id":"4d4553d893970092b711da4ea490384599583979","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-05-04 12:08:52.000000000","message":"Patch Set 42:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/1898 : FAILURE","accounts_in_message":[],"_revision_number":42},{"id":"abb423f4e44942972d56a11ae0399fcd9d3aeea4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 12:23:34.000000000","message":"Patch Set 42:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/42/check/DellEMC_XtremIO_FC/8e60d0b/DellEMC_XtremIO_FC/138 : SUCCESS in 55m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/42/check/DellEMC_XtremIO_iSCSI/51a0d5d/DellEMC_XtremIO_iSCSI/132 : SUCCESS in 1h 17m 17s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":42},{"id":"fe54e78e87ca7ad68e6373855d1bb3bcfcde6895","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-05-04 12:24:04.000000000","message":"Patch Set 42:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/42/check/DellEMC_Unity_FC/bab5b38/DellEMC_Unity_FC/110 : SUCCESS in 1h 10m 01s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/42/check/DellEMC_Unity_iSCSI/db06bcb/DellEMC_Unity_iSCSI/109 : SUCCESS in 1h 17m 55s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":42},{"id":"9b2c5a4445521201ea5f0870ac7229dd90981bac","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 12:24:31.000000000","message":"Patch Set 42:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-42-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"0e2895319cdc5c36f3d5f5f7effa1e051eb20d79","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-05-04 12:33:31.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/65/621465/42/check/DellEMC_SC_FC/9e0624a/DellEMC_SC_FC/33 : SUCCESS in 1h 07m 10s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/65/621465/42/check/DellEMC_SC_iSCSI/66c0d41/DellEMC_SC_iSCSI/32 : FAILURE in 1h 27m 14s\n\nLeave a comment with \u0027run-dell-emc-sc\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":42},{"id":"85b5b667332f2a33ee8668e0fa253b1d197f6587","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-04 12:37:05.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/42/upstream-check/cinder-cDOT-iSCSI/1308e78/ : FAILURE in 1h 30m 25s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/42/upstream-check/cinder-cDOT-NFS/277da14/ : FAILURE in 1h 29m 11s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/42/upstream-check/cinder-cDOT-FCP/591b77c/ : SUCCESS in 56m 11s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":42},{"id":"e73db8625bc8c6ee0ef17006de248b3ad0c85a8e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-04 12:45:14.000000000","message":"Patch Set 42:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"fe4b8ea28888fbe29a05f1d36b12c8ac8d6ba89c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-05-04 12:54:51.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/42/upstream-solidfire-check/cinder-solidfire-iSCSI/5b3f6a6/ : FAILURE in 1h 47m 45s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":42},{"id":"e21e85a7d41e2fa2942fb2282d59492107653d64","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-05-04 13:22:13.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/42/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/b2e4021 : SUCCESS in 2h 09m 04s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/42/check/3par-fc-driver-master-client-pip-ssa02-dsvm/6d23663 : SUCCESS in 2h 01m 38s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/42/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/c7a9c41 : SUCCESS in 2h 13m 39s","accounts_in_message":[],"_revision_number":42},{"id":"e975c148bdd73d0fab08a782b88ea5e0454b3398","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 13:59:48.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/65/621465/42/check/DellEMC_VNX_iSCSI/9a75277/DellEMC_VNX_iSCSI/44 : FAILURE in 2h 53m 32s\n- DellEMC_VNX_FC http://publiclogs.emc.com/65/621465/42/check/DellEMC_VNX_FC/518498b/DellEMC_VNX_FC/45 : SUCCESS in 1h 11m 09s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":42},{"id":"6e967099c4842286271836d47d067aed5ed32e63","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-04 14:30:29.000000000","message":"Patch Set 42:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"24d8ff0f313575be0c6a094966e815cbf5d40f83","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 19:26:05.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/65/621465/42/check/DellEMC_PowerMAX_FC/c46de4b/DellEMC_PowerMAX_FC/59 : SUCCESS in 2h 30m 21s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/65/621465/42/check/DellEMC_PowerMAX_iSCSI/1950727/DellEMC_PowerMAX_iSCSI/59 : FAILURE in 2h 03m 48s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":42},{"id":"ec9dadb4216ff554474cdcb19301bc7c05184101","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-07 19:51:36.000000000","message":"Patch Set 42: Code-Review-1\n\n(1 comment)\n\nPlease also remove the lvm driver from being defined in the cinder.conf on the RSD CI","accounts_in_message":[],"_revision_number":42},{"id":"8dfb9851eda0be9dcec2d510c35d44229f06f533","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-22 09:01:06.000000000","message":"Patch Set 42:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/42 : FAILURE","accounts_in_message":[],"_revision_number":42},{"id":"74ed6724cf0d96d05f4a641a01332c2d1735ce85","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-23 16:53:13.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"0dcafa9f9b280d8f9c6ecc1cc85b866ec5be4be1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-05-23 16:53:47.000000000","message":"Patch Set 43:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/43/check/dsvm-tempest-infinibox-fc/d07f291 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/43/check/dsvm-tempest-infinibox-iscsi/9cf3d7e : NOT_REGISTERED","accounts_in_message":[],"_revision_number":43},{"id":"6544bacb93fc2eb3e81c1b25102365c7aaaa6c37","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-23 17:40:14.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/43/check/dsvm-tempest-kvm/d87dfe6 : FAILURE in 39m 14s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/43/check/dsvm-tempest-exe/3206d15 : FAILURE in 34m 32s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":43},{"id":"30aab22495ea67fff5813b3eff5f1dfcc40b8ede","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-05-23 17:49:33.000000000","message":"Patch Set 43:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-05-23_09-57-13/ : FAILURE","accounts_in_message":[],"_revision_number":43},{"id":"cb71d371f7327b70c16795d29cade2fef29a3a99","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-23 17:59:54.000000000","message":"Patch Set 43:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/43/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/2d3119c : SUCCESS in 1h 04m 44s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/43/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/1bbfb1b : FAILURE in 45s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":43},{"id":"9ee86ef3f385a52b3c46743281b1e2e9c6a950dd","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-05-23 18:06:13.000000000","message":"Patch Set 43:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/65/621465/43/check/DellEMC_PowerMAX_FC/cb86135/DellEMC_PowerMAX_FC/164 : FAILURE in 25s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/65/621465/43/check/DellEMC_PowerMAX_iSCSI/80401b1/DellEMC_PowerMAX_iSCSI/175 : FAILURE in 20s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":43},{"id":"21a0891fd51b62c707474e5e081487a660df4e89","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-23 18:06:27.000000000","message":"Patch Set 43:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/43/check-not-vote/ext-cinder-zuul/9a15d17 : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":43},{"id":"cd60497d75f8e41be4737c727587e627448203fd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-05-23 18:23:06.000000000","message":"Patch Set 43:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/43 : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"3fd2eccad9cd3ba46c655d530aaa2d6b74596f1f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-23 18:25:46.000000000","message":"Patch Set 43:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/43/iscsi/ : FAILURE in 1h 30m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/43/smb/ : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":43},{"id":"d4e5887fc0cf72f86cfb45b35cc14d8862cb8bb6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-23 18:42:01.000000000","message":"Patch Set 43:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/43/check/DellEMC_XtremIO_FC/5e92aaf/DellEMC_XtremIO_FC/379 : SUCCESS in 53m 32s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/43/check/DellEMC_XtremIO_iSCSI/e4d36e2/DellEMC_XtremIO_iSCSI/350 : FAILURE in 1h 44m 22s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":43},{"id":"525853d7770d3f73d71ee75762b6612aa076c442","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-05-23 18:53:21.000000000","message":"Patch Set 43:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/43/check/fujitsu-eternus-dx-iscsi/4d6864d : SUCCESS in 1h 58m 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/43/check/fujitsu-eternus-dx-fc/8731f42 : SUCCESS in 1h 56m 42s","accounts_in_message":[],"_revision_number":43},{"id":"7d6b0e88bf98b651d516b5b11f4c646babd3ea36","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-05-23 23:15:40.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":43},{"id":"b9cb2732ea1afc9617053245bbfc610542ff55ed","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-23 23:19:02.000000000","message":"Patch Set 43:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/43/check/tempest-dsvm-full-xenial/2191e3b/ : FAILURE in 33m 16s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/43/check/devstack-plugin-ceph-dsvm-xenial/95219ac/ : SUCCESS in 1h 27m 55s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"b558dea4d793e4c1dfc169bb15b22c47154d8661","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 23:27:49.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/43/check/openstack-tox-lower-constraints/3f81bdd/ : SUCCESS in 10m 54s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/43/check/openstack-tox-pep8/cf1e60d/ : SUCCESS in 8m 04s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/43/check/openstack-tox-py27/811cc0a/ : SUCCESS in 7m 52s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/43/check/openstack-tox-py36/ec7ccc6/ : SUCCESS in 8m 18s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/43/check/openstack-tox-py37/73caa34/ : FAILURE in 8m 57s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/65/621465/43/check/openstack-tox-docs/1d2cf55/html/ : SUCCESS in 7m 58s\n- requirements-check http://logs.openstack.org/65/621465/43/check/requirements-check/9a796c0/ : SUCCESS in 2m 41s\n- tempest-full http://logs.openstack.org/65/621465/43/check/tempest-full/213f6ed/ : SUCCESS in 1h 45m 32s\n- neutron-grenade http://logs.openstack.org/65/621465/43/check/neutron-grenade/25d09d9/ : SUCCESS in 54m 59s\n- grenade-py3 http://logs.openstack.org/65/621465/43/check/grenade-py3/6015e55/ : SUCCESS in 48m 32s\n- tempest-full-py3 http://logs.openstack.org/65/621465/43/check/tempest-full-py3/6504e7e/ : SUCCESS in 1h 36m 00s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/43/check/build-openstack-releasenotes/de3cfd0/html/ : SUCCESS in 17m 31s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/43/check/cinder-tempest-dsvm-lvm-lio-barbican/47d6b57/ : SUCCESS in 1h 34m 41s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/43/check/cinder-tox-bandit-baseline/59bad48/ : FAILURE in 10m 57s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/43/check/openstack-tox-functional-py36/dbae511/ : SUCCESS in 5m 37s\n- cinder-rally-task http://logs.openstack.org/65/621465/43/check/cinder-rally-task/4a5a0ba/ : SUCCESS in 37m 42s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/43/check/openstack-tox-pylint/380b196/ : SUCCESS in 4m 12s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/43/check/devstack-plugin-ceph-tempest/b841ab1/ : SUCCESS in 1h 26m 23s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/43/check/openstack-tox-functional/e91d332/ : SUCCESS in 4m 56s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/43/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/e7696ee/ : SUCCESS in 1h 10m 41s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/43/check/legacy-tempest-dsvm-lvm-multibackend/d869913/ : SUCCESS in 1h 28m 23s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/43/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/cbc235c/ : SUCCESS in 1h 31m 45s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/43/check/tempest-slow-py3/9c3a339/ : SUCCESS in 2h 07m 37s","accounts_in_message":[],"_revision_number":43},{"id":"21a0303df423ffa894cb02349f41e9e0fabe6a6c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-05-23 23:28:29.000000000","message":"Patch Set 43:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/65/621465/43/check/DellEMC_PowerMAX_FC/b59b503/DellEMC_PowerMAX_FC/173 : FAILURE in 26s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/65/621465/43/check/DellEMC_PowerMAX_iSCSI/a299b91/DellEMC_PowerMAX_iSCSI/185 : FAILURE in 23s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":43},{"id":"a656257e2313e22584f9fb49a32751ac6fae4a49","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-05-23 23:52:36.000000000","message":"Patch Set 43:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-43/nedge_iscsi : FAILURE in 22m 36s","accounts_in_message":[],"_revision_number":43},{"id":"08643255b83fc8dac420c5d72635c539148ee2e6","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-05-24 02:13:06.000000000","message":"Patch Set 43:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/2085 : FAILURE","accounts_in_message":[],"_revision_number":43},{"id":"b5a91a387afcb6460fd3295a3592f98fba54ebc6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-05-24 02:16:23.000000000","message":"Patch Set 43:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/43/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a5d4812 : SUCCESS in 2h 04m 29s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/43/check/3par-fc-driver-master-client-pip-ssa02-dsvm/7f333c0 : SUCCESS in 2h 09m 31s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/43/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/e6c960c : FAILURE in 2h 08m 12s","accounts_in_message":[],"_revision_number":43},{"id":"e0f05a1a7294c67b5af5eb589c7ee5bce289a1b4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-24 04:38:54.000000000","message":"Patch Set 43:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/65/621465/43/check/DellEMC_VxFlexOS/e7f5b7d/DellEMC_VxFlexOS/187 : SUCCESS in 1h 34m 42s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":43},{"id":"4a5d4d06c976178187604e0dec4d3aa854ac071c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-24 06:52:43.000000000","message":"Patch Set 43:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/43/check/DellEMC_XtremIO_FC/3bc97b8/DellEMC_XtremIO_FC/387 : SUCCESS in 52m 59s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/43/check/DellEMC_XtremIO_iSCSI/584c7e4/DellEMC_XtremIO_iSCSI/358 : SUCCESS in 1h 15m 15s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":43},{"id":"b852cb81a4265f8923bcc58c74d63f5ed362813a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-05-24 08:10:31.000000000","message":"Patch Set 43:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/65/621465/43/check/DellEMC_VNX_iSCSI/e7c91f5/DellEMC_VNX_iSCSI/187 : FAILURE in 2h 56m 09s\n- DellEMC_VNX_FC http://publiclogs.emc.com/65/621465/43/check/DellEMC_VNX_FC/433168a/DellEMC_VNX_FC/190 : SUCCESS in 1h 24m 55s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":43},{"id":"ef5b8d7728254fc5ed4e4e56b86b49e346e67e1e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-05-24 08:40:47.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/43/check/NEC-ISCSI/4ddab3c : SUCCESS in 1h 59m 12s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/43/check/NEC-FC/441126e : SUCCESS in 1h 28m 31s","accounts_in_message":[],"_revision_number":43},{"id":"1f47112bcc1462c498a986ffc26ba4af590dd04e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-05-24 09:08:20.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/43/upstream-solidfire-check/cinder-solidfire-iSCSI/f22eb24/ : SUCCESS in 2h 51m 13s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":43},{"id":"0601d87d0b81034a5b2054a82d102cfd73bd07ec","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-05-24 09:12:16.000000000","message":"Patch Set 43:\n\nUpdate from the CI: We are currently fixing the CI after a site wide powerdown this week. Before the CI went down, I addressed hemna comment and removed LVM driver. We also added cinder-tempest-plugin.","accounts_in_message":[],"_revision_number":43},{"id":"e94b207bc612d6575fee4516b292c58396a39565","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-24 11:02:48.000000000","message":"Patch Set 43:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/43/check/dsvm-tempest-zadara-driver/4641aab : FAILURE in 1h 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"959e4ef16caaa9b007f4da2bfb08c9cf69ca2f38","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-24 13:59:16.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/43/upstream-check/cinder-cDOT-iSCSI/5c306d7/ : POST_FAILURE in 4h 05m 50s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/43/upstream-check/cinder-cDOT-NFS/c5c1d30/ : SUCCESS in 2h 18m 43s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/43/upstream-check/cinder-cDOT-FCP/33a602c/ : FAILURE in 2h 35m 34s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":43},{"id":"8f6566415da48dab6657cbe791aecb926d1b40dd","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-05-25 02:02:01.000000000","message":"Patch Set 43:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-43-iscsi : FAILURE","accounts_in_message":[],"_revision_number":43},{"id":"d0706a5f7486c22715fd311d57875613afcb0557","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-05-27 04:45:41.000000000","message":"Patch Set 43:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/13465 : SUCCESS in 1h 01m 27s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9766 : SUCCESS in 1h 00m 19s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/11486 : SUCCESS in 1h 20m 32s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/10208 : SUCCESS in 1h 12m 54s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/11112 : FAILURE in 55m 06s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1908 : FAILURE in 57s","accounts_in_message":[],"_revision_number":43},{"id":"b3859159443a5f10d0394b7fa10a4806e0698201","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-05-27 08:13:30.000000000","message":"Patch Set 42:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/13468 : FAILURE in 1h 01m 59s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9769 : SUCCESS in 1h 00m 26s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/11489 : SUCCESS in 1h 14m 13s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/10211 : SUCCESS in 1h 14m 33s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/11115 : FAILURE in 57m 52s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1911 : FAILURE in 52s","accounts_in_message":[],"_revision_number":42},{"id":"94eb52644b0dbaa5a364fe70f303da2712d296a6","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-27 13:02:07.000000000","message":"Patch Set 43:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/43 : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"f32eb2c6b9488af8c19d3b0024e46c7726541cd4","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2019-05-27 13:09:12.000000000","message":"Patch Set 43: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":43},{"id":"ed25bc0d1ddf0cb2485bbc1c3f622338bd1c48a0","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-05-27 13:26:20.000000000","message":"Patch Set 43:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":43},{"id":"c5274668b8570293aa98f88ed722641e6a728156","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-27 16:53:31.000000000","message":"Patch Set 43:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/43 : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"5d040dec34b2f6364c33da79cb56417d956ab794","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-05-27 17:38:35.000000000","message":"Patch Set 43:\n\nrecheck","accounts_in_message":[],"_revision_number":43},{"id":"76288f10fcfa507a4e3863ff179e569467a23da0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-27 19:10:51.000000000","message":"Patch Set 43:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/43/check/tempest-dsvm-full-xenial/d401a3d/ : FAILURE in 1h 27m 11s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/43/check/devstack-plugin-ceph-dsvm-xenial/a930f6d/ : FAILURE in 1h 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"cadb35a642cf31fa62e3db408fd11b71677a7a4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 19:48:35.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/43/check/openstack-tox-lower-constraints/e0d0081/ : SUCCESS in 10m 05s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/43/check/openstack-tox-pep8/3c39bbc/ : SUCCESS in 7m 44s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/43/check/openstack-tox-py27/221b359/ : SUCCESS in 8m 16s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/43/check/openstack-tox-py36/d2cb05c/ : SUCCESS in 8m 11s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/43/check/openstack-tox-py37/d407d9a/ : FAILURE in 8m 05s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/65/621465/43/check/openstack-tox-docs/31b3793/html/ : SUCCESS in 7m 45s\n- requirements-check http://logs.openstack.org/65/621465/43/check/requirements-check/8088dcd/ : SUCCESS in 3m 04s\n- tempest-full http://logs.openstack.org/65/621465/43/check/tempest-full/cf701b3/ : SUCCESS in 1h 35m 51s\n- neutron-grenade http://logs.openstack.org/65/621465/43/check/neutron-grenade/ed48e4b/ : SUCCESS in 46m 03s\n- grenade-py3 http://logs.openstack.org/65/621465/43/check/grenade-py3/47441de/ : SUCCESS in 56m 07s\n- tempest-full-py3 http://logs.openstack.org/65/621465/43/check/tempest-full-py3/ff3e776/ : SUCCESS in 1h 45m 22s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/43/check/build-openstack-releasenotes/1ff7771/html/ : SUCCESS in 13m 10s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/43/check/cinder-tempest-dsvm-lvm-lio-barbican/7a4d737/ : FAILURE in 1h 43m 53s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/43/check/cinder-tox-bandit-baseline/c220661/ : FAILURE in 11m 31s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/43/check/openstack-tox-functional-py36/1b24a9d/ : SUCCESS in 5m 14s\n- cinder-rally-task http://logs.openstack.org/65/621465/43/check/cinder-rally-task/991f5af/ : SUCCESS in 44m 17s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/43/check/openstack-tox-pylint/69e59dd/ : SUCCESS in 5m 58s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/43/check/devstack-plugin-ceph-tempest/2f744e8/ : SUCCESS in 1h 38m 25s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/43/check/openstack-tox-functional/157ed50/ : SUCCESS in 4m 38s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/43/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/c9f8928/ : SUCCESS in 1h 00m 44s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/43/check/legacy-tempest-dsvm-lvm-multibackend/e170637/ : SUCCESS in 1h 18m 18s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/43/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/fb2efc7/ : SUCCESS in 1h 42m 32s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/43/check/tempest-slow-py3/b405c2b/ : SUCCESS in 2h 08m 23s","accounts_in_message":[],"_revision_number":43},{"id":"aefe6fd11241cf9dafed226f26d067cb33d57329","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-27 20:12:24.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/43/iscsi/ : SUCCESS in 50m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/43/smb/ : SUCCESS in 48m 54s","accounts_in_message":[],"_revision_number":43},{"id":"d1b027e747767dbe9f49cc4e3797cf982d9fb6ca","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-27 20:47:16.000000000","message":"Patch Set 43:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/43 : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"80c7f977fac53b00c0b4a0438e2d6922aaf88c8c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-27 21:38:40.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/43/check/dsvm-tempest-kvm/1daddc1 : FAILURE in 39m 16s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/43/check/dsvm-tempest-exe/f94c8ac : FAILURE in 34m 36s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":43},{"id":"95b662ac4d016bb55c6a2e721e649949ba7d77cf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-27 21:42:47.000000000","message":"Patch Set 43:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/43/check-not-vote/ext-cinder-zuul/b960053 : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":43},{"id":"55cfdf5d1a43fc66234e8187bac97d3df042a86b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-27 22:24:26.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/43/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/05a0cbc : SUCCESS in 54m 59s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/43/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/8fa5552 : SUCCESS in 1h 05m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":43},{"id":"05439b6b64c16e13aa32734b7e92bee0cc966365","author":{"_account_id":15961,"name":"lisali","email":"xiaoyan.li@intel.com","username":"lisali"},"date":"2019-05-28 01:03:13.000000000","message":"Patch Set 43:\n\n(3 comments)","accounts_in_message":[],"_revision_number":43},{"id":"fb50578936bb980ec219636ee8d162db33dc36f6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-05-28 01:22:31.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/43/check/dsvm-tempest-cisco-zonemanager-job/28511e0/ : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":43},{"id":"380902ecbfa4a5d5e06e39b7e8a7a8e5258c36fc","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-28 05:10:24.000000000","message":"Patch Set 43:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/43/check/dsvm-tempest-zadara-driver/2c76662 : FAILURE in 1h 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"7d8645a2528ade35af191de6f4b60d845bfd80d4","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2019-05-28 06:13:58.000000000","message":"Patch Set 43:\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"3a5398d9fe665d0fbfafcae523210c7819be707d","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-05-28 06:42:12.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/43/check/kaminario-dsvm-tempest-full-iscsi/e01475a : SUCCESS in 2h 16m 03s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/43/check/kaminario-dsvm-tempest-full-FC/c51db64 : SUCCESS in 1h 53m 02s","accounts_in_message":[],"_revision_number":43},{"id":"4b9f4d81ff429c2e063e3d06e78ee27e9dc5b225","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-28 21:41:09.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"05b9cf185240765ec24a6b6d108768fe3195adcf","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-05-28 21:41:51.000000000","message":"Patch Set 44:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/65/621465/44/check/DellEMC_PowerMAX_FC/e799753/DellEMC_PowerMAX_FC/207 : FAILURE in 26s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/65/621465/44/check/DellEMC_PowerMAX_iSCSI/69a95d4/DellEMC_PowerMAX_iSCSI/219 : FAILURE in 20s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":44},{"id":"543cf7998cd9ea510b8edc13fb2963a2de27a283","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-28 22:14:46.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/44/check/dsvm-tempest-kvm/2bff65f : FAILURE in 32m 58s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/44/check/dsvm-tempest-exe/f57a37c : FAILURE in 22m 46s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":44},{"id":"992d4bb0a093430885e4310e87f6df737c36496d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-28 22:26:03.000000000","message":"Patch Set 44:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/44/check-not-vote/ext-cinder-zuul/5b319e6 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":44},{"id":"d8975be027cb02f1dbbea349edc0f8d26d0c67d2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-28 22:35:03.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/44/iscsi/ : SUCCESS in 53m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/44/smb/ : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":44},{"id":"69d000377b15400281ffea520c8923194f0ffd08","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-05-28 22:53:00.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/44/check/dsvm-tempest-infinibox-fc/d4bfd2b : SUCCESS in 54m 48s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/44/check/dsvm-tempest-infinibox-iscsi/34e6b3d : SUCCESS in 1h 11m 07s","accounts_in_message":[],"_revision_number":44},{"id":"1ebde442cfedbf8ba70dd5126d537c93cccf078b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-28 22:57:24.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/44/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/62da57f : SUCCESS in 52m 06s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/44/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/788e8ed : SUCCESS in 1h 15m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":44},{"id":"a1b2a7769bbb84c20b4dd97455f14b37766c422d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-05-28 23:11:16.000000000","message":"Patch Set 44:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"f8763ca936921a756dcc39c1f52ec77f927a7c98","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-05-28 23:13:19.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/44/check/dsvm-tempest-cisco-zonemanager-job/71a66f0/ : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":44},{"id":"2f06f765aeb8b3c37cf4fa8cd7fd16b66640e4f4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-05-28 23:16:07.000000000","message":"Patch Set 44:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/44/check/fujitsu-eternus-dx-iscsi/15c86df : SUCCESS in 1h 29m 31s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/44/check/fujitsu-eternus-dx-fc/0765886 : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":44},{"id":"34ea05ea6a2a6572127c5a24e0459d074a6dfeeb","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-28 23:16:36.000000000","message":"Patch Set 44:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/44/check/DellEMC_XtremIO_FC/d98a9c3/DellEMC_XtremIO_FC/425 : SUCCESS in 53m 03s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/44/check/DellEMC_XtremIO_iSCSI/19c718e/DellEMC_XtremIO_iSCSI/395 : SUCCESS in 1h 22m 59s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":44},{"id":"9c6e7c1d99eded2a28627bd4441fa4b68ef0727a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-28 23:18:05.000000000","message":"Patch Set 44:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/44/check/tempest-dsvm-full-xenial/16bfe3b/ : SUCCESS in 1h 25m 07s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/44/check/devstack-plugin-ceph-dsvm-xenial/d767946/ : SUCCESS in 1h 36m 26s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"8a0e3fdeecdb5f484e5fbe1077a35e9551d78aba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 00:16:56.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/44/check/openstack-tox-lower-constraints/41a1f55/ : SUCCESS in 9m 56s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/44/check/openstack-tox-pep8/b3af5eb/ : SUCCESS in 7m 01s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/44/check/openstack-tox-py27/f04c446/ : SUCCESS in 8m 26s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/44/check/openstack-tox-py36/530a852/ : SUCCESS in 8m 13s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/44/check/openstack-tox-py37/8ebe8e2/ : FAILURE in 8m 52s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/65/621465/44/check/openstack-tox-docs/5e3866e/html/ : SUCCESS in 7m 55s\n- requirements-check http://logs.openstack.org/65/621465/44/check/requirements-check/d3e6372/ : SUCCESS in 2m 46s\n- tempest-full http://logs.openstack.org/65/621465/44/check/tempest-full/4af5995/ : SUCCESS in 1h 42m 07s\n- neutron-grenade http://logs.openstack.org/65/621465/44/check/neutron-grenade/599c66c/ : SUCCESS in 59m 30s\n- grenade-py3 http://logs.openstack.org/65/621465/44/check/grenade-py3/838c71a/ : SUCCESS in 1h 01m 48s\n- tempest-full-py3 http://logs.openstack.org/65/621465/44/check/tempest-full-py3/70687a2/ : SUCCESS in 1h 33m 10s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/44/check/build-openstack-releasenotes/8eecbca/html/ : SUCCESS in 19m 05s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/44/check/cinder-tempest-dsvm-lvm-lio-barbican/6658ead/ : SUCCESS in 1h 44m 02s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/44/check/cinder-tox-bandit-baseline/3f9abd8/ : FAILURE in 11m 01s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/44/check/openstack-tox-functional-py36/a3defdf/ : SUCCESS in 4m 46s\n- cinder-rally-task http://logs.openstack.org/65/621465/44/check/cinder-rally-task/d9a7fe7/ : SUCCESS in 37m 53s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/44/check/openstack-tox-pylint/ce99668/ : SUCCESS in 4m 45s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/44/check/devstack-plugin-ceph-tempest/ed4f249/ : SUCCESS in 1h 47m 55s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/44/check/openstack-tox-functional/68cced4/ : SUCCESS in 5m 11s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/44/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/e498f0c/ : SUCCESS in 1h 09m 33s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/44/check/legacy-tempest-dsvm-lvm-multibackend/b93c61f/ : SUCCESS in 1h 43m 43s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/44/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/62bd3bb/ : SUCCESS in 1h 50m 31s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/44/check/tempest-slow-py3/b339430/ : FAILURE in 2h 24m 47s","accounts_in_message":[],"_revision_number":44},{"id":"2731e056725290cd71c262c5fd6ce2cd104abad0","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-29 00:20:54.000000000","message":"Patch Set 44:\n\n(3 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":44},{"id":"45f30778d763fefb86809f8946b2b86b57c46c62","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 00:37:53.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/44/check/NEC-ISCSI/b83badc : SUCCESS in 1h 55m 07s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/44/check/NEC-FC/3007929 : SUCCESS in 1h 54m 15s","accounts_in_message":[],"_revision_number":44},{"id":"a4dfc3bd76978f655a779367fa7566f8fce9b20b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-29 00:49:32.000000000","message":"Patch Set 44:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"15279be1adb106e79b17b39bbfc634229149b24a","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-05-29 02:11:43.000000000","message":"Patch Set 44:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-05-28_14-41-19/ : FAILURE","accounts_in_message":[],"_revision_number":44},{"id":"da593bbec5f68b257abc13c24d3fb112da75391e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-29 04:03:23.000000000","message":"Patch Set 44:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/44/check/dsvm-tempest-zadara-driver/6378c86 : FAILURE in 1h 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"68c6518c3b2eff4359fffeea1c491dac4c184f88","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-29 05:17:07.000000000","message":"Patch Set 44:\n\nrecheck","accounts_in_message":[],"_revision_number":44},{"id":"a05e89a4fa3654f0567991fcfa2c516159b7e8d5","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-29 05:46:07.000000000","message":"Patch Set 44:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"1be70e21bfc2711401d29410c92d9f5c334cd627","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 05:57:11.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/44/check/dsvm-tempest-kvm/02421b3 : FAILURE in 33m 41s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/44/check/dsvm-tempest-exe/35c6ea6 : FAILURE in 38m 53s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":44},{"id":"8125cd317c783ff0da77a205158374d2ee253231","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-29 06:29:26.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/44/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/7f18bc6 : SUCCESS in 56m 53s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/44/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/11585db : SUCCESS in 1h 11m 55s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":44},{"id":"56a03f8d6404e1d9543dcc44c86e99c081995833","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-29 07:04:49.000000000","message":"Patch Set 44:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/44/check/dsvm-tempest-zadara-driver/aae52c4 : FAILURE in 1h 47m 27s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"fa309ccabb1339f01200993d28239159d5324a28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 07:52:11.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/44/check/openstack-tox-lower-constraints/58e0e74/ : SUCCESS in 10m 22s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/44/check/openstack-tox-pep8/3363fa8/ : SUCCESS in 6m 24s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/44/check/openstack-tox-py27/05ca6c4/ : SUCCESS in 7m 29s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/44/check/openstack-tox-py36/9820ce4/ : SUCCESS in 7m 56s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/44/check/openstack-tox-py37/766a9e2/ : FAILURE in 8m 28s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/65/621465/44/check/openstack-tox-docs/a8c30e6/html/ : SUCCESS in 8m 13s\n- requirements-check http://logs.openstack.org/65/621465/44/check/requirements-check/39ca929/ : SUCCESS in 2m 24s\n- tempest-full http://logs.openstack.org/65/621465/44/check/tempest-full/5115da4/ : SUCCESS in 1h 48m 33s\n- neutron-grenade http://logs.openstack.org/65/621465/44/check/neutron-grenade/c78d561/ : SUCCESS in 53m 27s\n- grenade-py3 http://logs.openstack.org/65/621465/44/check/grenade-py3/d566517/ : SUCCESS in 1h 03m 17s\n- tempest-full-py3 http://logs.openstack.org/65/621465/44/check/tempest-full-py3/45a9221/ : SUCCESS in 1h 41m 27s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/44/check/build-openstack-releasenotes/0cdc678/html/ : SUCCESS in 13m 18s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/44/check/cinder-tempest-dsvm-lvm-lio-barbican/7682a07/ : SUCCESS in 2h 03m 58s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/44/check/cinder-tox-bandit-baseline/9893f2e/ : FAILURE in 10m 31s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/44/check/openstack-tox-functional-py36/ebec871/ : SUCCESS in 4m 38s\n- cinder-rally-task http://logs.openstack.org/65/621465/44/check/cinder-rally-task/9529455/ : SUCCESS in 37m 16s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/44/check/openstack-tox-pylint/0655ad2/ : SUCCESS in 4m 54s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/44/check/devstack-plugin-ceph-tempest/daa5fb5/ : SUCCESS in 1h 26m 12s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/44/check/openstack-tox-functional/7098df7/ : SUCCESS in 4m 30s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/44/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/0307006/ : SUCCESS in 1h 13m 03s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/44/check/legacy-tempest-dsvm-lvm-multibackend/77669af/ : FAILURE in 1h 27m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/44/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/3bd6a9f/ : SUCCESS in 1h 33m 41s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/44/check/tempest-slow-py3/170b61d/ : SUCCESS in 2h 15m 11s","accounts_in_message":[],"_revision_number":44},{"id":"16f59a88cde45ddaa529ff456d0fe98b6e1b4fe6","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 08:05:57.000000000","message":"Patch Set 44:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-65-621465-44/index.html : FAILURE \n You can rerun this CI by commenting run-Datera","accounts_in_message":[],"_revision_number":44},{"id":"22ba8f41acc362bfcb33e7e9a4896e267296143d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 08:15:28.000000000","message":"Patch Set 44:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/65/621465/44/check/DellEMC_VNX_iSCSI/c99861d/DellEMC_VNX_iSCSI/208 : FAILURE in 22m 59s\n- DellEMC_VNX_FC http://publiclogs.emc.com/65/621465/44/check/DellEMC_VNX_FC/24841aa/DellEMC_VNX_FC/211 : FAILURE in 20m 06s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":44},{"id":"abaff03de37ac6e875b4892c3306b9e5e389f41b","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 08:19:56.000000000","message":"Patch Set 44:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-65-621465-44/index.html : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"4467cf98bf8f1ce9c049c75809171715f18a6a49","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-05-29 08:20:42.000000000","message":"Patch Set 44:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-44/nedge_iscsi : FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":44},{"id":"461ef1fb638e4864afdf0037769dc2c47136d950","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-05-29 08:35:18.000000000","message":"Patch Set 44:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-44/ns5_iscsi : FAILURE in 40m 58s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-44/ns5_nfs : FAILURE in 24m 40s","accounts_in_message":[],"_revision_number":44},{"id":"7fe6ae3ef70367169106a628b1c90afffc9cacca","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-05-29 08:52:15.000000000","message":"Patch Set 44:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/2105 : FAILURE","accounts_in_message":[],"_revision_number":44},{"id":"b14a895198f4280a3c1f7a691b5c547f7ecaa627","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 09:07:11.000000000","message":"Patch Set 44:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-44-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"03e6a85d46346feed48ef971891c5c4fbb892e3e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 09:28:53.000000000","message":"Patch Set 44:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/65/621465/44/check/DellEMC_VxFlexOS/ee2d44f/DellEMC_VxFlexOS/208 : SUCCESS in 1h 36m 18s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":44},{"id":"ed6b0190ce6f510bd01c92669247944fdb50e12b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-05-29 09:34:36.000000000","message":"Patch Set 44:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/44/Lenovo-FC : SUCCESS in 33:13.78\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/44/Lenovo-iSCSI : SUCCESS in 34:13.75","accounts_in_message":[],"_revision_number":44},{"id":"fb06bbfda45a2b88293b5f1432acb28c20293009","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2019-05-29 09:35:31.000000000","message":"Patch Set 44:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- xiv-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/xiv-tempest-dsvm-full-iscsi/13476 : SUCCESS in 57m 05s\n- ds8k-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/ds8k-tempest-dsvm-full-fc/9777 : FAILURE in 2h 49m 00s\n- storwize-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-iscsi/11497 : SUCCESS in 1h 35m 27s\n- storwize-tempest-dsvm-full-fc http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/storwize-tempest-dsvm-full-fc/10219 : SUCCESS in 1h 06m 45s\n- flashsystem-tempest-dsvm-full-iscsi http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/flashsystem-tempest-dsvm-full-iscsi/11123 : FAILURE in 1h 06m 59s\n- gpfs-cinder-tempest-dsvm-full-file http://dal05.objectstorage.softlayer.net/v1/AUTH_0bd27569-6310-44fc-8a8d-36a112cac4ec/IBM-STORAGE-CI/gpfs-cinder-tempest-dsvm-full-file/1919 : FAILURE in 55s","accounts_in_message":[],"_revision_number":44},{"id":"9a13ca98a48ab50a9fb483788f3872c876752b7b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 09:35:33.000000000","message":"Patch Set 44:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/44/check/DellEMC_XtremIO_FC/745d154/DellEMC_XtremIO_FC/427 : SUCCESS in 57m 29s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/44/check/DellEMC_XtremIO_iSCSI/1338036/DellEMC_XtremIO_iSCSI/397 : SUCCESS in 1h 42m 58s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":44},{"id":"b132cf273154861eaf211bd154a190eb76531d65","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-05-29 09:54:35.000000000","message":"Patch Set 44:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/44/HPMSA-FC : SUCCESS in 43:29.28\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/44/HPMSA-iSCSI : SUCCESS in 43:14.80","accounts_in_message":[],"_revision_number":44},{"id":"d7e8410d6cf0a2ae10411179461f1bdec9540b1a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-05-29 10:03:51.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/44/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/48ae1ba : SUCCESS in 2h 09m 01s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/44/check/3par-fc-driver-master-client-pip-ssa02-dsvm/5957004 : SUCCESS in 2h 05m 21s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/44/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/bb348a6 : SUCCESS in 2h 00m 06s","accounts_in_message":[],"_revision_number":44},{"id":"3dc5c6c378533d8c39772128d0ff3516931427e9","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-05-29 10:08:15.000000000","message":"Patch Set 44:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-44 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI","accounts_in_message":[],"_revision_number":44},{"id":"7b8761f62f3f7b067531edbc9b2c74ac0404672d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-29 11:01:55.000000000","message":"Patch Set 44:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/44/iscsi/ : FAILURE in 14m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/44/smb/ : FAILURE in 14m 45s","accounts_in_message":[],"_revision_number":44},{"id":"57d1710502280ce9e05e9da0defc3661bbf466ae","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-05-29 12:23:49.000000000","message":"Patch Set 44:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-44 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI","accounts_in_message":[],"_revision_number":44},{"id":"bb81a2914ae06157a9e4e860ba1a0556f547a334","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-05-29 13:24:34.000000000","message":"Patch Set 44:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/44/check/kaminario-dsvm-tempest-full-iscsi/471f681 : SUCCESS in 2h 22m 44s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/44/check/kaminario-dsvm-tempest-full-FC/8150f0d : FAILURE in 1h 31m 53s","accounts_in_message":[],"_revision_number":44},{"id":"5b4a19ce0042dc33c95bf36780ea436dc015da15","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 13:41:32.000000000","message":"Patch Set 44:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/65/621465/44/check/DellEMC_PowerMAX_FC/f5fd525/DellEMC_PowerMAX_FC/209 : FAILURE in 5h 48m 49s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/65/621465/44/check/DellEMC_PowerMAX_iSCSI/59c113b/DellEMC_PowerMAX_iSCSI/221 : FAILURE in 5h 49m 06s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":44},{"id":"a4ea3f76648027244e0c0d043955b17e7bee078a","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-05-29 15:29:36.000000000","message":"Patch Set 44:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/2106 : FAILURE","accounts_in_message":[],"_revision_number":44},{"id":"648cd6def616e6068d8e90193a85f60e7048e045","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-05-29 16:46:58.000000000","message":"Patch Set 44:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-44 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI","accounts_in_message":[],"_revision_number":44},{"id":"56d381c5ac3ccdff36aafe31bbac9dca4cfa3e09","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-05-29 23:53:54.000000000","message":"Topic changed from bp/nvme-volume-driver to train-drivers","accounts_in_message":[],"_revision_number":44},{"id":"201acd76e7901da53f3fe73484ac5d185b78bb3b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-30 01:20:05.000000000","message":"Patch Set 44:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/44/check/tempest-dsvm-full-xenial/f9bdbf0/ : SUCCESS in 1h 22m 54s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/44/check/devstack-plugin-ceph-dsvm-xenial/8615043/ : SUCCESS in 1h 35m 42s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"ba5471c4b56f517acd420448839cf1b151eb6c6f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-05-30 05:23:09.000000000","message":"Patch Set 44:\n\nrecheck","accounts_in_message":[],"_revision_number":44},{"id":"dcfd898826a5017d223ef26b7c2a22d8064b9fce","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-05-30 05:39:00.000000000","message":"Patch Set 44:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/65/621465/44/check-cinder/Cinder-ISER-ISCSI/5b32345 : FAILURE in 1h 01m 35s\n- Cinder-ISER-LIO http://13.74.249.42/65/621465/44/check-cinder/Cinder-ISER-LIO/a59261d : SUCCESS in 1h 07m 43s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":44},{"id":"69fae291292f2b381bad207fec589d2cb7aaf697","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-30 06:00:17.000000000","message":"Patch Set 44:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/44/check-not-vote/ext-cinder-zuul/6a741a8 : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":44},{"id":"906f915f29fa1160362bdecd906ca24a64f819d4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-30 06:00:24.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/44/check/dsvm-tempest-kvm/1608231 : FAILURE in 36m 42s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/44/check/dsvm-tempest-exe/fa8540d : FAILURE in 34m 06s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":44},{"id":"d090aacca15dc6d4fc11d46f085824b831226df8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-05-30 07:07:38.000000000","message":"Patch Set 44:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/44/check/dsvm-tempest-zadara-driver/8309257 : FAILURE in 1h 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"f6f7a8db8ddb091cd15e0ff397eb13358a7d427d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-30 07:14:18.000000000","message":"Patch Set 44:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/44/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/e99f105 : FAILURE in 1h 50m 50s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/44/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/ea9f326 : SUCCESS in 1h 14m 27s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":44},{"id":"d85a6c93f948bf5638747ac6f035b531e42f3dae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 07:30:12.000000000","message":"Patch Set 44:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/44/check/openstack-tox-lower-constraints/af95690/ : SUCCESS in 9m 46s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/44/check/openstack-tox-pep8/2e63bda/ : SUCCESS in 6m 55s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/44/check/openstack-tox-py27/b181b0d/ : SUCCESS in 8m 26s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/44/check/openstack-tox-py36/e28be48/ : SUCCESS in 7m 44s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/44/check/openstack-tox-py37/adb76a8/ : SUCCESS in 9m 42s\n- openstack-tox-docs http://logs.openstack.org/65/621465/44/check/openstack-tox-docs/3468bca/html/ : SUCCESS in 7m 06s\n- requirements-check http://logs.openstack.org/65/621465/44/check/requirements-check/1581ea1/ : SUCCESS in 3m 01s\n- tempest-full http://logs.openstack.org/65/621465/44/check/tempest-full/845356a/ : SUCCESS in 1h 44m 49s\n- neutron-grenade http://logs.openstack.org/65/621465/44/check/neutron-grenade/946ebc8/ : SUCCESS in 57m 35s\n- grenade-py3 http://logs.openstack.org/65/621465/44/check/grenade-py3/5edbd4a/ : SUCCESS in 47m 46s\n- tempest-full-py3 http://logs.openstack.org/65/621465/44/check/tempest-full-py3/2c36d05/ : SUCCESS in 1h 30m 57s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/44/check/build-openstack-releasenotes/790c3bc/html/ : SUCCESS in 14m 54s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/44/check/cinder-tempest-dsvm-lvm-lio-barbican/0d6f535/ : SUCCESS in 1h 31m 01s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/44/check/cinder-tox-bandit-baseline/6072499/ : FAILURE in 9m 51s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/44/check/openstack-tox-functional-py36/6680f16/ : SUCCESS in 5m 59s\n- cinder-rally-task http://logs.openstack.org/65/621465/44/check/cinder-rally-task/6a0f85b/ : SUCCESS in 35m 58s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/44/check/openstack-tox-pylint/c24d224/ : SUCCESS in 4m 33s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/44/check/devstack-plugin-ceph-tempest/caaffa7/ : SUCCESS in 1h 39m 09s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/44/check/openstack-tox-functional/63ff9fe/ : SUCCESS in 6m 29s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/44/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/479bae9/ : SUCCESS in 1h 06m 16s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/44/check/legacy-tempest-dsvm-lvm-multibackend/4224d56/ : SUCCESS in 1h 31m 43s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/44/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/b98b14d/ : SUCCESS in 1h 26m 08s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/44/check/tempest-slow-py3/afc4f08/ : SUCCESS in 2h 00m 34s","accounts_in_message":[],"_revision_number":44},{"id":"b072e781904f20313b1ce9a1998214cb8e2b9d4a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-05-30 09:01:28.000000000","message":"Patch Set 44:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/65/621465/44/check/DellEMC_VNX_iSCSI/798ee7f/DellEMC_VNX_iSCSI/216 : SUCCESS in 1h 30m 58s\n- DellEMC_VNX_FC http://publiclogs.emc.com/65/621465/44/check/DellEMC_VNX_FC/fbd2844/DellEMC_VNX_FC/219 : SUCCESS in 1h 03m 59s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":44},{"id":"0b58a386e898c59e32ad46bfa1560f5f522bd6a2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-05-30 09:37:40.000000000","message":"Patch Set 44:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/44/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/78348bb : FAILURE in 3m 54s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/44/check/3par-fc-driver-master-client-pip-ssa02-dsvm/5f029f1 : SUCCESS in 2h 04m 55s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/44/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/ea0e31a : FAILURE in 31m 59s","accounts_in_message":[],"_revision_number":44},{"id":"393229133e26c833b993367bf2886198daafeacb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-05-30 10:12:42.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/44/check/dsvm-tempest-cisco-zonemanager-job/1a52e7d/ : SUCCESS in 1h 04m 07s","accounts_in_message":[],"_revision_number":44},{"id":"d347fafb308160edb0b0cd4e2cfbf222e9bb98e0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-30 10:38:55.000000000","message":"Patch Set 44:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/65/621465/44/check/DellEMC_VxFlexOS/a8a96fc/DellEMC_VxFlexOS/216 : SUCCESS in 1h 34m 39s\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":44},{"id":"de6b049150c0104c1a4a15f9da034b3ebcf31800","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-05-30 11:36:34.000000000","message":"Patch Set 44:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/44/check/DellEMC_XtremIO_FC/2fd2e26/DellEMC_XtremIO_FC/445 : SUCCESS in 56m 12s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/44/check/DellEMC_XtremIO_iSCSI/c097275/DellEMC_XtremIO_iSCSI/414 : FAILURE in 15m 26s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":44},{"id":"5ce75aa0ceb088c3e704d8cdcbbfd0d91aabf3b8","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-30 12:13:23.000000000","message":"Patch Set 44: Code-Review+2","accounts_in_message":[],"_revision_number":44},{"id":"bb5968414678b64d2bdbf30bc8026f091345bff8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-05-30 12:24:34.000000000","message":"Patch Set 44:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/44/iscsi/ : FAILURE in 14m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/44/smb/ : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":44},{"id":"34caf62522c9ca8ebd6eeb51e2d3a485485f8976","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-05-30 13:01:36.000000000","message":"Patch Set 44:\n\nrun netapp-ci","accounts_in_message":[],"_revision_number":44},{"id":"a3fbef884bf8eaa04199b1d1eac9cb716fd26b65","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-30 13:58:05.000000000","message":"Patch Set 44:\n\nLooks like all comments have been addressed.\n\nOnly issue I now see is the CI is running using Python 3.5 but needs to use Python 3.7.","accounts_in_message":[],"_revision_number":44},{"id":"8a153cf91791ab7b9a5c4d6fd845203973f78b53","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-05-30 14:01:42.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/44/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/30abbab : SUCCESS in 2h 06m 21s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/44/check/3par-fc-driver-master-client-pip-ssa02-dsvm/c21eaa9 : SUCCESS in 1h 54m 24s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/44/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/6cf9e59 : SUCCESS in 2h 02m 32s","accounts_in_message":[],"_revision_number":44},{"id":"e7de6a7dad8d373bbf5011ed52eb9f82cb016fb4","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-05-30 14:29:31.000000000","message":"Patch Set 44:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":44},{"id":"30014f277841c9b1a4aafb8bb0374bb83ea21933","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-05-30 15:22:26.000000000","message":"Patch Set 44:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":44},{"id":"c51c8628e2a79db6981c93d569cd5d9147bbeb05","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-30 15:31:08.000000000","message":"Patch Set 44:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/44 : FAILURE","accounts_in_message":[],"_revision_number":44},{"id":"bf84c18ac0b200b1657bce23b2b848d7061dea84","author":{"_account_id":29637,"name":"OSNEXUS Quantastor CI","email":"eng.openstack@osnexus.com","username":"quantastor-ci"},"date":"2019-05-30 15:45:34.000000000","message":"Patch Set 44:\n\n* cinder-quantastor-dsvm-tempest http://openstack-ci.osnexus.net.s3.us-east-1.amazonaws.com/index.html?prefix\u003drefs-changes-65-621465-44 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quantastor_CI","accounts_in_message":[],"_revision_number":44},{"id":"dfabf80ba0120f1ef4da918730450e7c463ff76b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-05-30 15:50:06.000000000","message":"Patch Set 44:\n\nrun netapp-ci","accounts_in_message":[],"_revision_number":44},{"id":"a00d140767c0c160792393e1b341cbdef18366c1","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-05-30 15:51:10.000000000","message":"Patch Set 44:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":44},{"id":"0e073b4b6a0fa514a2fc9d9afd8350b4f18fc938","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-05-30 16:43:10.000000000","message":"Patch Set 44:\n\nrun netapp-ci","accounts_in_message":[],"_revision_number":44},{"id":"089e34451b37d2bcd8dd2cd9f31fef609aedd18e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-05-30 18:38:34.000000000","message":"Patch Set 44:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"16c03c76ba3f93a2722577cf186dcbd64fb666e2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-05-30 18:40:13.000000000","message":"Patch Set 44:\n\nBuild failed.  To recheck use \u0027run-storpoolci\u0027.  For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/StorPool_CI\n\n- dsvm-tempest-storpool http://logs.ci-openstack.storpool.com/65/621465/44/check/dsvm-tempest-storpool/66e006c : EXCEPTION","accounts_in_message":[],"_revision_number":44},{"id":"5aad95db12d0c4639e45160e627cd519312f5b5c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-30 19:27:35.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/44/upstream-check/cinder-cDOT-iSCSI/7d47dbf/ : SUCCESS in 1h 35m 04s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/44/upstream-check/cinder-cDOT-NFS/342c082/ : SUCCESS in 1h 43m 05s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/44/upstream-check/cinder-cDOT-FCP/427fb06/ : FAILURE in 1h 38m 03s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":44},{"id":"7c8b5532ef83c2b2ad23e25fbb77da5c12a92fc3","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-05-30 21:16:30.000000000","message":"Patch Set 44:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/44/check/kaminario-dsvm-tempest-full-iscsi/4b68096 : SUCCESS in 1h 59m 37s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/44/check/kaminario-dsvm-tempest-full-FC/6a596e9 : FAILURE in 1h 25m 17s","accounts_in_message":[],"_revision_number":44},{"id":"b82247e7a7c84022dce03445f321d94f022820ae","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-31 09:19:44.000000000","message":"Patch Set 44:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/44/check/tempest-dsvm-full-xenial/f2f63f1/ : SUCCESS in 1h 23m 29s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/44/check/devstack-plugin-ceph-dsvm-xenial/586fa87/ : FAILURE in 1h 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"6301e081d0c270451d3503caa5fa4766cf2c1bd1","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-06-06 10:51:13.000000000","message":"Patch Set 44:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":44},{"id":"df73d8f62f556b0494a5fd64e4393fb27718be1e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-06-06 11:22:54.000000000","message":"Patch Set 44:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"689e07a25eac15243bf65602d426e8d3b01d392b","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-06-06 11:55:09.000000000","message":"Patch Set 44:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":44},{"id":"8b97a86776739ad61ce7ddb51a2cfc920dc086b1","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-06-06 14:32:13.000000000","message":"Patch Set 44:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":44},{"id":"363ae61cf619b4a181ad4189c497af7e5346be8a","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-06-06 14:43:37.000000000","message":"Patch Set 44:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"c981a4e07ad0b87120177a11e8b659168fd0f2c8","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-06-06 14:57:10.000000000","message":"Patch Set 44:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/44 : FAILURE","accounts_in_message":[],"_revision_number":44},{"id":"cf3e60d495a7361e06829b4846e3d66c7e26b525","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-06-06 15:01:43.000000000","message":"Patch Set 44:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/44 : FAILURE","accounts_in_message":[],"_revision_number":44},{"id":"f88c66e187d6029731ebdba45e178d22b637428e","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-06-06 21:54:04.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"97ad28ce2d012e292767f6ddb33b5d2f6d18f52d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-06 21:54:33.000000000","message":"Patch Set 45:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/65/621465/45/check/dsvm-tempest-infinibox-fc/8f47c73 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/65/621465/45/check/dsvm-tempest-infinibox-iscsi/440f0f9 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":45},{"id":"9ad8b4a93cb2b15bb13b7c1ff53fafa19d93f4fb","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-06-06 21:55:23.000000000","message":"Topic changed from bp/nvme-volume-driver to train-drivers","accounts_in_message":[],"_revision_number":45},{"id":"8e7dcbaf23634c477195d31af324e08ff65900c9","author":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"date":"2019-06-06 21:58:10.000000000","message":"Patch Set 45:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":45},{"id":"f71600f688065cd8061da9293b86817ad2346e1c","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-06-06 22:01:33.000000000","message":"Patch Set 45:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1460 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1513 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1674 : FAILURE","accounts_in_message":[],"_revision_number":45},{"id":"fdc6d79ad969ccf9be2f7381e209ec3f766ae288","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-06 22:14:55.000000000","message":"Patch Set 45:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/621465/45/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/ad5e303 : EXCEPTION\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/621465/45/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/ce53b76 : EXCEPTION\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":45},{"id":"c6855cb31bd158883a425e56c14ba930c73d875a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-06 22:31:16.000000000","message":"Patch Set 45:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/45/check-not-vote/ext-cinder-zuul/92a917f : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":45},{"id":"ceaec3e617b6b29506932fee40477ff449e3180f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-06 22:53:57.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/45/check/dsvm-tempest-kvm/b4f20c8 : FAILURE in 33m 46s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/45/check/dsvm-tempest-exe/4b91ed2 : FAILURE in 38m 48s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":45},{"id":"ab2550174b62b7949f086db1c0de1b317c918613","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-06 23:00:27.000000000","message":"Patch Set 45:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/45/iscsi/ : FAILURE in 1h 05m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/45/smb/ : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":45},{"id":"98e29253612ae06b7c34cd4a5d8cf206bb5d5efb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-06 23:21:59.000000000","message":"Patch Set 45:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/45/check/tempest-dsvm-full-xenial/dfdef4f/ : SUCCESS in 1h 27m 11s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/45/check/devstack-plugin-ceph-dsvm-xenial/54524eb/ : SUCCESS in 1h 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"82cc8ce03538f194d3f0845f7284b55f51e22195","author":{"_account_id":15296,"name":"Brocade Openstack CI","email":"Brocade-Openstack-CI@Brocade.com","username":"Brocade-Openstack-CI","tags":["SERVICE_USER"]},"date":"2019-06-06 23:26:10.000000000","message":"Patch Set 45:\n\nBuild failed. \r\n* brocade_fc_zone_driver http://192.19.229.191:8000/build_logs/2019-06-06_15-33-28/ : FAILURE","accounts_in_message":[],"_revision_number":45},{"id":"086a55ca51d56fef887b0a091723eb37ad51883a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-06 23:30:39.000000000","message":"Patch Set 45:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/65/621465/45 : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"0ceaec6a8e03ff0f5f51d5ece5f5d6e4f7756106","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-06-06 23:40:39.000000000","message":"Patch Set 45:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/65/621465/45/check/fujitsu-eternus-dx-iscsi/7ad90db : SUCCESS in 1h 44m 23s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/65/621465/45/check/fujitsu-eternus-dx-fc/d2d1218 : SUCCESS in 1h 43m 15s","accounts_in_message":[],"_revision_number":45},{"id":"978c9130a4a078c4908de8c7a3ee4289456116b0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-07 01:20:45.000000000","message":"Patch Set 45:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/45/check/DellEMC_Unity_FC/a3ce498/DellEMC_Unity_FC/393 : SUCCESS in 1h 10m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/45/check/DellEMC_Unity_iSCSI/4b20beb/DellEMC_Unity_iSCSI/364 : SUCCESS in 1h 21m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":45},{"id":"d7200da2ae327d23a30316a51c673ea8ff14e686","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-06-07 01:35:23.000000000","message":"Patch Set 45:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/45 : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"541f5df3f883aa8033a5cc858b1d700cf7daf6b2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-07 03:24:40.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/45/check/dsvm-tempest-cisco-zonemanager-job/1ccb604/ : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":45},{"id":"b4bf1b995c54bf3faa719efb198fd00f5820d273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 04:51:57.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/45/check/openstack-tox-lower-constraints/e76f3d7/ : SUCCESS in 11m 50s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/45/check/openstack-tox-pep8/b290e8b/ : SUCCESS in 8m 04s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/45/check/openstack-tox-py27/f54d604/ : SUCCESS in 9m 00s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/45/check/openstack-tox-py36/5ef41f2/ : SUCCESS in 8m 33s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/45/check/openstack-tox-py37/04e18f2/ : SUCCESS in 9m 18s\n- openstack-tox-docs http://logs.openstack.org/65/621465/45/check/openstack-tox-docs/e9061e3/html/ : SUCCESS in 8m 18s\n- requirements-check http://logs.openstack.org/65/621465/45/check/requirements-check/cd42463/ : SUCCESS in 2m 49s\n- tempest-full http://logs.openstack.org/65/621465/45/check/tempest-full/749557b/ : SUCCESS in 1h 41m 10s\n- neutron-grenade http://logs.openstack.org/65/621465/45/check/neutron-grenade/0f22f5f/ : SUCCESS in 1h 01m 20s\n- grenade-py3 http://logs.openstack.org/65/621465/45/check/grenade-py3/8cdcbb4/ : SUCCESS in 57m 23s\n- tempest-full-py3 http://logs.openstack.org/65/621465/45/check/tempest-full-py3/7142f09/ : SUCCESS in 1h 40m 11s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/45/check/build-openstack-releasenotes/f521267/html/ : SUCCESS in 14m 28s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/45/check/cinder-tempest-dsvm-lvm-lio-barbican/335d10f/ : SUCCESS in 2h 04m 16s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/45/check/cinder-tox-bandit-baseline/e3f70bd/ : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/45/check/openstack-tox-functional-py36/8b80b2b/ : SUCCESS in 4m 48s\n- cinder-rally-task http://logs.openstack.org/65/621465/45/check/cinder-rally-task/b164514/ : SUCCESS in 46m 09s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/45/check/openstack-tox-pylint/ec0f08a/ : SUCCESS in 4m 33s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/45/check/devstack-plugin-ceph-tempest/a188e3b/ : SUCCESS in 1h 13m 38s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/45/check/openstack-tox-functional/cc52a0d/ : SUCCESS in 4m 53s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/45/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/5efa4af/ : SUCCESS in 1h 02m 50s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/45/check/legacy-tempest-dsvm-lvm-multibackend/12108fe/ : SUCCESS in 1h 27m 54s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/45/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/92eadf6/ : FAILURE in 1h 24m 12s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/45/check/tempest-slow-py3/3a27679/ : SUCCESS in 2h 18m 39s","accounts_in_message":[],"_revision_number":45},{"id":"efb8d497541f781c9c52f218dfe79ca49dc68466","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 05:09:46.000000000","message":"Patch Set 45:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-65-621465-45/index.html : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"0caebf1c28eba0102f9c4ae1bca54c8493e29f1b","author":{"_account_id":16834,"name":"Nexenta Edge CI","email":"nexentaedge@gmail.com","username":"nexentaedge","tags":["SERVICE_USER"]},"date":"2019-06-07 05:20:48.000000000","message":"Patch Set 45:\n\n* nexenta-dsvm-volume-nedge_iscsi http://140.174.232.106/refs-changes-65-621465-45/nedge_iscsi : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":45},{"id":"93dd8452d2aab4e359b73f9ba2cf7a91b8624e5f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-06-07 05:32:45.000000000","message":"Patch Set 45:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-45/ns5_iscsi : SUCCESS in 37m 34s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-45/ns5_nfs : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":45},{"id":"8cdb09d756e3450fd6e067d39aa2bbfd9932443c","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-06-07 05:46:36.000000000","message":"Patch Set 45:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/65/2206 : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"b052a124e12017d75aed069bfa773c4583703f2d","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 06:07:57.000000000","message":"Patch Set 45:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-45-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"48944b982c3e109e98b4569d4f55fbb64e9e6733","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 06:08:27.000000000","message":"Patch Set 45:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/65/621465/45/check/DellEMC_VxFlexOS/1bd80ec/DellEMC_VxFlexOS/280 : SUCCESS in 1h 16m 10s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":45},{"id":"cd6a9260ca033aae1652aa42205c529311ef06dc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-07 06:18:29.000000000","message":"Patch Set 45:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/65/621465/45/check/DellEMC_Unity_FC/f283eb9/DellEMC_Unity_FC/396 : SUCCESS in 1h 18m 36s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/65/621465/45/check/DellEMC_Unity_iSCSI/a6cf12d/DellEMC_Unity_iSCSI/367 : SUCCESS in 1h 26m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":45},{"id":"48486f298ef19c8d2eddcc4d654206d62021ad1a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-06-07 06:28:21.000000000","message":"Patch Set 45:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/65/621465/45/check/DellEMC_SC_FC/c374d35/DellEMC_SC_FC/118 : SUCCESS in 1h 01m 51s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/65/621465/45/check/DellEMC_SC_iSCSI/90d7d20/DellEMC_SC_iSCSI/115 : SUCCESS in 1h 36m 07s\n\nLeave a comment with \u0027run-dellemc-sc\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_SC_Series_CI","accounts_in_message":[],"_revision_number":45},{"id":"fb59303d45355497a7dee148f59e9139e94b9f6d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 06:30:13.000000000","message":"Patch Set 45:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/65/621465/45/check/DellEMC_VNX_iSCSI/0ab6f76/DellEMC_VNX_iSCSI/283 : SUCCESS in 1h 37m 58s\n- DellEMC_VNX_FC http://publiclogs.emc.com/65/621465/45/check/DellEMC_VNX_FC/3eaad0a/DellEMC_VNX_FC/290 : SUCCESS in 1h 00m 32s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":45},{"id":"87d77bf3b62c5fdd8f6d1dcb9165b0628e435ca3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-06-07 06:34:40.000000000","message":"Patch Set 45:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/45/Lenovo-FC : SUCCESS in 31:59.00\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/45/Lenovo-iSCSI : SUCCESS in 33:49.80","accounts_in_message":[],"_revision_number":45},{"id":"2bb00012025ea3ee1b980ff5a1b64de7f4c08124","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 06:44:53.000000000","message":"Patch Set 45:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/65/621465/45/check/DellEMC_XtremIO_FC/d0b7091/DellEMC_XtremIO_FC/547 : SUCCESS in 51m 11s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/65/621465/45/check/DellEMC_XtremIO_iSCSI/95a4398/DellEMC_XtremIO_iSCSI/509 : FAILURE in 1h 45m 32s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_XtremIO_CI","accounts_in_message":[],"_revision_number":45},{"id":"7d28e65c58b23d2fb7da1c481dc9154d5d909a1c","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-06-07 06:45:03.000000000","message":"Patch Set 45:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/45/HPMSA-FC : SUCCESS in 37:49.94\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/45/HPMSA-iSCSI : SUCCESS in 38:46.83","accounts_in_message":[],"_revision_number":45},{"id":"5a647bb37dcb184e196d1b66d1aa3ec7b4cd282e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-07 07:25:15.000000000","message":"Patch Set 45:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/45/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/eb8de9b : FAILURE in 2h 30m 43s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/45/check/3par-fc-driver-master-client-pip-ssa02-dsvm/0d81961 : FAILURE in 2h 11m 26s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/45/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/5a67999 : SUCCESS in 2h 05m 03s","accounts_in_message":[],"_revision_number":45},{"id":"b15c12a20b38c5367136118ab517f5817d3c0402","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-06-07 07:27:47.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/45/upstream-solidfire-check/cinder-solidfire-iSCSI/639ff2c/ : SUCCESS in 2h 34m 38s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":45},{"id":"ff3b1d428bf258156e93f93f881c848b5a757e7f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-07 08:02:45.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/65/621465/45/upstream-check/cinder-cDOT-iSCSI/d6067a3/ : FAILURE in 3h 10m 03s\n- cinder-cDOT-NFS http://13.56.179.158/logs/65/621465/45/upstream-check/cinder-cDOT-NFS/ceb53e4/ : SUCCESS in 2h 03m 42s\n- cinder-cDOT-FCP http://13.56.179.158/logs/65/621465/45/upstream-check/cinder-cDOT-FCP/ec6d2b7/ : FAILURE in 2h 25m 16s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":45},{"id":"cb3b578030df6c916449235dd2f5ff41ab433824","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 08:21:18.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/65/621465/45/check/NEC-ISCSI/06920c6 : SUCCESS in 1h 47m 45s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/65/621465/45/check/NEC-FC/a524c65 : SUCCESS in 1h 48m 46s","accounts_in_message":[],"_revision_number":45},{"id":"82d83beeec5e8175c92ef6e5adc13f54e7cc8037","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-07 18:30:50.000000000","message":"Patch Set 45: Code-Review+2 Review-Priority+1\n\nLooks like all comments have been addressed and third party CI is running appropriate tests under Python 3.7.\n\nThis driver needs to be added to the support matrix, but that can be done in a follow up. See https://review.opendev.org/#/c/580757/","accounts_in_message":[],"_revision_number":45},{"id":"ae8b311009523f26b8d3afc4d66c92eb7ec4ab41","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-06-07 21:58:48.000000000","message":"Patch Set 45:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/621465/45 : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"06807f35509ed20e54c69eb2f90e1cafcec76103","author":{"_account_id":16258,"name":"nikesh","email":"nike.niec@gmail.com","username":"nikem"},"date":"2019-06-12 05:37:09.000000000","message":"Patch Set 45:\n\nrun-Kaminario K2 CI","accounts_in_message":[],"_revision_number":45},{"id":"1790f579c011432fa66a6d6910b1eab944811bb0","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-06-12 07:35:13.000000000","message":"Patch Set 45:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/45/check/kaminario-dsvm-tempest-full-iscsi/b591f8e : FAILURE in 1h 57m 28s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/45/check/kaminario-dsvm-tempest-full-FC/1935e78 : SUCCESS in 1h 48m 46s","accounts_in_message":[],"_revision_number":45},{"id":"c4da3e71510edb3a412bae34c4a658e1f5944c4d","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-06-12 16:48:58.000000000","message":"Patch Set 45: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":45},{"id":"c4a084023968d8a8d8dc44008ca0b933130f3028","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 16:49:25.000000000","message":"Patch Set 45: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":45},{"id":"e02e01d4b94754aae2e8724c1d70c293b3dd1554","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 18:36:54.000000000","message":"Patch Set 45: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/45/gate/openstack-tox-lower-constraints/79c8f3c/ : SUCCESS in 11m 27s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/45/gate/openstack-tox-pep8/d626ee3/ : TIMED_OUT in 30m 58s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/45/gate/openstack-tox-py27/b48490f/ : SUCCESS in 10m 27s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/45/gate/openstack-tox-py36/9a24756/ : SUCCESS in 8m 00s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/45/gate/openstack-tox-py37/7c0a66c/ : TIMED_OUT in 40m 54s\n- openstack-tox-docs http://logs.openstack.org/65/621465/45/gate/openstack-tox-docs/a367736/html/ : SUCCESS in 7m 43s\n- requirements-check http://logs.openstack.org/65/621465/45/gate/requirements-check/f7b75ef/ : SUCCESS in 3m 11s\n- tempest-full http://logs.openstack.org/65/621465/45/gate/tempest-full/bd87b57/ : FAILURE in 30m 10s\n- neutron-grenade http://logs.openstack.org/65/621465/45/gate/neutron-grenade/728f621/ : SUCCESS in 1h 01m 17s\n- grenade-py3 http://logs.openstack.org/65/621465/45/gate/grenade-py3/88ebeea/ : SUCCESS in 1h 03m 02s\n- tempest-full-py3 http://logs.openstack.org/65/621465/45/gate/tempest-full-py3/2f46aa2/ : SUCCESS in 1h 31m 45s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/45/gate/build-openstack-releasenotes/c6ef42f/html/ : SUCCESS in 14m 36s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/45/gate/legacy-grenade-dsvm-cinder-mn-sub-volbak/a3d3fca/ : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":45},{"id":"e3757375791ce1619e825217fc9911a389df391e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-12 19:04:02.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"783ca4889dc1478798306132291b35bc25aa6a6d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-12 19:48:03.000000000","message":"Patch Set 45:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/65/621465/45/check-not-vote/ext-cinder-zuul/9df297a : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":45},{"id":"5cbeba7d1aa05c480892e3456f2c3b570e4402e5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 19:49:08.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/45/check/dsvm-tempest-kvm/4c1ab4a : FAILURE in 41m 39s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/65/621465/45/check/dsvm-tempest-exe/21f20dc : FAILURE in 43m 36s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":45},{"id":"f51a3cf6bf9fb65905de02c39a822c756070d40b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-12 19:58:13.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/621465/45/iscsi/ : SUCCESS in 53m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/621465/45/smb/ : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":45},{"id":"feb42c6d8e8b928dd555de7e7403834e5b0087c4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-12 20:36:36.000000000","message":"Patch Set 45:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/45/check/tempest-dsvm-full-xenial/ddfde1b/ : FAILURE in 1h 23m 25s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/65/621465/45/check/devstack-plugin-ceph-dsvm-xenial/1025e3b/ : FAILURE in 1h 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"1fb283c80bd16d2efaf18bea49aa20d8e63de224","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-06-12 22:00:35.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline). Use run-ZadaraStorage VPSA CI to recheck. For information, see https://wiki.openstack.org/wiki/ThirdPartySystems/ZadaraStorage_VPSA_CI.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/65/621465/45/check/dsvm-tempest-zadara-driver/ab0f6c0 : FAILURE in 1h 52m 45s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"ed5b143ac2a90bcfb8f700cac398e9ef19b78502","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-12 22:46:47.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/65/621465/45/check/dsvm-tempest-cisco-zonemanager-job/f4d03de/ : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":45},{"id":"6cb1dc287439d2856a7c4861ddd9f20ab2535e9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 03:53:25.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/45/check/openstack-tox-lower-constraints/16f5be5/ : SUCCESS in 10m 54s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/45/check/openstack-tox-pep8/38e732f/ : SUCCESS in 8m 11s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/45/check/openstack-tox-py27/16b0083/ : SUCCESS in 8m 20s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/45/check/openstack-tox-py36/8cf7153/ : SUCCESS in 8m 09s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/45/check/openstack-tox-py37/b5e3308/ : SUCCESS in 8m 08s\n- openstack-tox-docs http://logs.openstack.org/65/621465/45/check/openstack-tox-docs/26b2540/html/ : SUCCESS in 8m 08s\n- requirements-check http://logs.openstack.org/65/621465/45/check/requirements-check/bdbec7a/ : SUCCESS in 3m 12s\n- tempest-full http://logs.openstack.org/65/621465/45/check/tempest-full/0fe529e/ : SUCCESS in 1h 48m 23s\n- neutron-grenade http://logs.openstack.org/65/621465/45/check/neutron-grenade/abc9cd1/ : SUCCESS in 54m 01s\n- grenade-py3 http://logs.openstack.org/65/621465/45/check/grenade-py3/5d11d6e/ : SUCCESS in 1h 02m 45s\n- tempest-full-py3 http://logs.openstack.org/65/621465/45/check/tempest-full-py3/7079aed/ : SUCCESS in 1h 38m 57s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/45/check/build-openstack-releasenotes/7310a87/html/ : SUCCESS in 18m 36s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/65/621465/45/check/cinder-tempest-dsvm-lvm-lio-barbican/d028503/ : SUCCESS in 2h 04m 49s\n- cinder-tox-bandit-baseline http://logs.openstack.org/65/621465/45/check/cinder-tox-bandit-baseline/6b91991/ : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/65/621465/45/check/openstack-tox-functional-py36/9eae521/ : SUCCESS in 5m 43s\n- cinder-rally-task http://logs.openstack.org/65/621465/45/check/cinder-rally-task/b661ff7/ : SUCCESS in 45m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/65/621465/45/check/openstack-tox-pylint/f1a9e49/ : SUCCESS in 4m 44s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/65/621465/45/check/devstack-plugin-ceph-tempest/8cc6dba/ : SUCCESS in 1h 37m 08s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/621465/45/check/openstack-tox-functional/26aa244/ : SUCCESS in 6m 14s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/45/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/f2c4399/ : SUCCESS in 1h 06m 46s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/65/621465/45/check/legacy-tempest-dsvm-lvm-multibackend/7329859/ : SUCCESS in 1h 26m 56s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/65/621465/45/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/7d3e993/ : SUCCESS in 1h 23m 58s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/65/621465/45/check/tempest-slow-py3/dd81c11/ : SUCCESS in 2h 14m 47s","accounts_in_message":[],"_revision_number":45},{"id":"dfc0b81b06b2a41e0feae0c53ea508601fe6141d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 03:53:58.000000000","message":"Patch Set 45: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":45},{"id":"1bb74250ce16deb06f31693d73d1614d835d19ac","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-06-13 03:56:30.000000000","message":"Patch Set 45:\n\n* nexenta-dsvm-volume-ns5_iscsi http://140.174.232.106/refs-changes-65-621465-45/ns5_iscsi : FAILURE in 0m 9s\n* nexenta-dsvm-volume-ns5_nfs http://140.174.232.106/refs-changes-65-621465-45/ns5_nfs : FAILURE in 0m 9s","accounts_in_message":[],"_revision_number":45},{"id":"cb59d5d6ff7267cc9b125eb788d74200df042575","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-06-13 05:11:48.000000000","message":"Patch Set 45:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-65-621465-45-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"c6a4490a7319cf396928eac7b7f277b2451d4161","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-13 05:13:57.000000000","message":"Patch Set 45:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-621465-45 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":45},{"id":"1ca7804ff62cb0163196cb411e5f7be482060ed7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-06-13 05:17:07.000000000","message":"Patch Set 45:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/65/621465/45/check/DellEMC_VxFlexOS/5a8cf35/DellEMC_VxFlexOS/309 : SUCCESS in 1h 23m 23s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":45},{"id":"d00079127e0f3de3261f821c62c947ba0dfccb90","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-06-13 05:57:09.000000000","message":"Patch Set 45:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/65/621465/45/Lenovo-FC : SUCCESS in 32:12.63\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/65/621465/45/Lenovo-iSCSI : SUCCESS in 33:37.81","accounts_in_message":[],"_revision_number":45},{"id":"4a61286fe6a14df5dc172f787ea5bb1a4a11008b","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-06-13 06:09:35.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/65/621465/45/check/kaminario-dsvm-tempest-full-iscsi/435d43e : SUCCESS in 1h 43m 47s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/65/621465/45/check/kaminario-dsvm-tempest-full-FC/d3ebe27 : SUCCESS in 1h 51m 32s","accounts_in_message":[],"_revision_number":45},{"id":"a0f04e41f6f259afb6d3a1f53d33023aa74f2534","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-13 06:11:17.000000000","message":"Patch Set 45:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/45/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c307294 : FAILURE in 2h 10m 15s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/65/621465/45/check/3par-fc-driver-master-client-pip-ssa02-dsvm/baafa1e : FAILURE in 2h 05m 26s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/65/621465/45/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/3595163 : SUCCESS in 2h 05m 06s","accounts_in_message":[],"_revision_number":45},{"id":"d9b74bd31b3bfee6dfb168e5a75fa5e1297676c4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-06-13 06:16:49.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/65/621465/45/upstream-solidfire-check/cinder-solidfire-iSCSI/e893d4e/ : FAILURE in 16m 37s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":45},{"id":"5b89d7afb173b721d58df55f0306961e47a6be59","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-06-13 06:43:51.000000000","message":"Patch Set 45:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/65/621465/45/HPMSA-FC : SUCCESS in 37:37.15\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/65/621465/45/HPMSA-iSCSI : SUCCESS in 37:53.98","accounts_in_message":[],"_revision_number":45},{"id":"0226737a9e7e3ac3b3a6c8a7a1bbf1a595723c27","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-06-13 07:08:17.000000000","message":"Patch Set 45:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/65/621465/45/check/DellEMC_SC_FC/832b602/DellEMC_SC_FC/160 : SUCCESS in 1h 13m 54s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/65/621465/45/check/DellEMC_SC_iSCSI/6db6472/DellEMC_SC_iSCSI/154 : FAILURE in 2h 51m 58s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":45},{"id":"060cacd748d3cc68fb9c50c86f69722f96aa04e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 07:50:06.000000000","message":"Patch Set 45: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/621465/45/gate/openstack-tox-lower-constraints/741e1e1/ : SUCCESS in 10m 56s\n- openstack-tox-pep8 http://logs.openstack.org/65/621465/45/gate/openstack-tox-pep8/2a393a8/ : SUCCESS in 8m 47s\n- openstack-tox-py27 http://logs.openstack.org/65/621465/45/gate/openstack-tox-py27/b54812a/ : SUCCESS in 11m 43s\n- openstack-tox-py36 http://logs.openstack.org/65/621465/45/gate/openstack-tox-py36/3e6339e/ : SUCCESS in 10m 19s\n- openstack-tox-py37 http://logs.openstack.org/65/621465/45/gate/openstack-tox-py37/39cd415/ : SUCCESS in 8m 59s\n- openstack-tox-docs http://logs.openstack.org/65/621465/45/gate/openstack-tox-docs/a5ee981/html/ : SUCCESS in 8m 07s\n- requirements-check http://logs.openstack.org/65/621465/45/gate/requirements-check/81bbbde/ : SUCCESS in 3m 21s\n- tempest-full http://logs.openstack.org/65/621465/45/gate/tempest-full/1db5a77/ : SUCCESS in 1h 38m 53s\n- neutron-grenade http://logs.openstack.org/65/621465/45/gate/neutron-grenade/0af8779/ : SUCCESS in 59m 47s\n- grenade-py3 http://logs.openstack.org/65/621465/45/gate/grenade-py3/746cc45/ : SUCCESS in 52m 16s\n- tempest-full-py3 http://logs.openstack.org/65/621465/45/gate/tempest-full-py3/5e391fe/ : SUCCESS in 1h 32m 37s\n- build-openstack-releasenotes http://logs.openstack.org/65/621465/45/gate/build-openstack-releasenotes/1e0ce3e/html/ : SUCCESS in 17m 36s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/65/621465/45/gate/legacy-grenade-dsvm-cinder-mn-sub-volbak/f38c11e/ : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":45},{"id":"8fff9dada5b6202bc69e617570a508a24af7ffc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 07:50:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":45},{"id":"e03df39c82751e0ece8b566e30b90d11d1271936","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-06-13 10:19:15.000000000","message":"Patch Set 45:\n\nIntel RSD CI is down for the next 24 hrs due to migration.","accounts_in_message":[],"_revision_number":45},{"id":"44e7d5d0549902c54f5c94b3ab7de60867c4c26a","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-06-13 11:03:48.000000000","message":"Patch Set 45:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":45},{"id":"def92d944e1d6de6c6abd6e5daa76fe53d863588","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-06-13 11:08:41.000000000","message":"Patch Set 45:\n\nIntel RSD CI migration complete. Cached test will now be run.","accounts_in_message":[],"_revision_number":45},{"id":"890a534459a69d7dd5b65566372623cb0b1d3dd0","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-06-14 10:00:21.000000000","message":"Patch Set 45:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/65/621465/45/check/synology-dsm-cinder-iscsi/2e53f04 : FAILURE in 27m 51s","accounts_in_message":[],"_revision_number":45},{"id":"6d9c98f551f47598068296d527af06ef72dde5b2","author":{"_account_id":29855,"name":"Martin Kennelly","email":"martin.kennelly@intel.com","username":"martinkennelly"},"date":"2019-06-14 11:55:07.000000000","message":"Patch Set 45:\n\nrun-Intel RSD CI","accounts_in_message":[],"_revision_number":45}],"current_revision_number":45,"current_revision":"b92b80241c86d74d71298077b3357989a9188b43","revisions":{"79b6302e08f2112f3dc51d759808ccb83b2fb301":{"kind":"REWORK","_number":1,"created":"2018-12-03 02:18:06.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/1"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nImplemented:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n\nTODO:\n- unit tests\n- copy volume to image\n- copy image to volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79b6302e08f2112f3dc51d759808ccb83b2fb301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79b6302e08f2112f3dc51d759808ccb83b2fb301"}]},"branch":"refs/heads/master"},"acb8c09a244f5158d1a35d2b77140553213b446b":{"kind":"REWORK","_number":2,"created":"2018-12-04 02:17:25.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/2"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-04 08:57:59.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nImplemented:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n\nTODO:\n- unit tests\n- copy volume to image\n- copy image to volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/acb8c09a244f5158d1a35d2b77140553213b446b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/acb8c09a244f5158d1a35d2b77140553213b446b"}]},"branch":"refs/heads/master"},"bb04d67abf8d99105a7fc4fc8b5442e43bc03485":{"kind":"REWORK","_number":3,"created":"2018-12-04 07:05:40.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/3"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-04 13:46:44.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nImplemented:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n\nTODO:\n- releasenote\n- unit tests\n- copy volume to image\n- copy image to volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb04d67abf8d99105a7fc4fc8b5442e43bc03485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb04d67abf8d99105a7fc4fc8b5442e43bc03485"}]},"branch":"refs/heads/master"},"afb928bbd13c6eb88a2ebb4a4f7882a41fa92469":{"kind":"REWORK","_number":4,"created":"2018-12-04 16:02:54.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/4"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-04 22:44:41.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nImplemented:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n\nTODO:\n- releasenote\n- unit tests\n- copy volume to image\n- copy image to volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afb928bbd13c6eb88a2ebb4a4f7882a41fa92469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afb928bbd13c6eb88a2ebb4a4f7882a41fa92469"}]},"branch":"refs/heads/master"},"d270ffcba58e6dddb6db59e54690da4dd5afc2ae":{"kind":"REWORK","_number":5,"created":"2018-12-05 08:58:53.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/5"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-05 15:40:33.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nImplemented:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n\nTODO:\n- releasenote\n- unit tests\n- copy volume to image\n- copy image to volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d270ffcba58e6dddb6db59e54690da4dd5afc2ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d270ffcba58e6dddb6db59e54690da4dd5afc2ae"}]},"branch":"refs/heads/master"},"80f088aa5e3b057e80011dfb29a68a34424b3085":{"kind":"REWORK","_number":6,"created":"2018-12-06 02:00:24.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/6"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-06 08:41:19.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nImplemented:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n\nTODO:\n- releasenote\n- unit tests\n- copy volume to image\n- copy image to volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80f088aa5e3b057e80011dfb29a68a34424b3085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80f088aa5e3b057e80011dfb29a68a34424b3085"}]},"branch":"refs/heads/master"},"dc73774ce977ea8800bddddeafdde908feaf0219":{"kind":"REWORK","_number":7,"created":"2018-12-06 06:01:20.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/7"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-06 12:39:51.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nImplemented:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n\nTODO:\n- releasenote\n- unit tests\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib volume inplace extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc73774ce977ea8800bddddeafdde908feaf0219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc73774ce977ea8800bddddeafdde908feaf0219"}]},"branch":"refs/heads/master"},"7708fdf305272222a26c16e872c560eab0834082":{"kind":"REWORK","_number":8,"created":"2018-12-07 03:11:27.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/8"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-07 09:52:32.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nImplemented:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n\nTODO:\n- releasenote\n- unit tests\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7708fdf305272222a26c16e872c560eab0834082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7708fdf305272222a26c16e872c560eab0834082"}]},"branch":"refs/heads/master"},"096b85e038ba4470aade43dad74f828a557f1fb9":{"kind":"REWORK","_number":9,"created":"2018-12-07 06:43:00.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/9"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-07 13:23:46.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/096b85e038ba4470aade43dad74f828a557f1fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/096b85e038ba4470aade43dad74f828a557f1fb9"}]},"branch":"refs/heads/master"},"5bc1dc98f1aa72d36d23087eebe721cd7b4f2d66":{"kind":"REWORK","_number":10,"created":"2018-12-07 06:45:52.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/10"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-07 13:27:06.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc1dc98f1aa72d36d23087eebe721cd7b4f2d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc1dc98f1aa72d36d23087eebe721cd7b4f2d66"}]},"branch":"refs/heads/master"},"df24286b910b996458a7bfc348db34c7f1877dba":{"kind":"REWORK","_number":11,"created":"2018-12-10 03:13:03.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/11"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-10 09:53:59.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df24286b910b996458a7bfc348db34c7f1877dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df24286b910b996458a7bfc348db34c7f1877dba"}]},"branch":"refs/heads/master"},"1e59b076dff8870c5fd685071684e8c1810fc90a":{"kind":"REWORK","_number":12,"created":"2018-12-10 14:54:19.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/12"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-10 21:35:24.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e59b076dff8870c5fd685071684e8c1810fc90a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e59b076dff8870c5fd685071684e8c1810fc90a"}]},"branch":"refs/heads/master"},"7f3773b5789a35e804075dcca4738512e220ce58":{"kind":"REWORK","_number":13,"created":"2018-12-10 15:02:37.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/13"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-10 21:44:02.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f3773b5789a35e804075dcca4738512e220ce58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f3773b5789a35e804075dcca4738512e220ce58"}]},"branch":"refs/heads/master"},"72c22ea9241ebf1312bc0aa5b34e0ccc428a0bb7":{"kind":"REWORK","_number":14,"created":"2018-12-12 05:10:36.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/14"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-12 11:51:11.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests (WIP)\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72c22ea9241ebf1312bc0aa5b34e0ccc428a0bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72c22ea9241ebf1312bc0aa5b34e0ccc428a0bb7"}]},"branch":"refs/heads/master"},"85a3d61b4601f42a2d2c2256bffc336ef977525c":{"kind":"REWORK","_number":15,"created":"2018-12-12 08:13:46.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/15"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-12 14:53:34.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests (WIP)\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85a3d61b4601f42a2d2c2256bffc336ef977525c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85a3d61b4601f42a2d2c2256bffc336ef977525c"}]},"branch":"refs/heads/master"},"7121d0045745584a4a9bc0bd569b37dd32852cff":{"kind":"REWORK","_number":16,"created":"2018-12-13 08:57:30.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/16"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-13 15:38:29.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests (WIP)\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7121d0045745584a4a9bc0bd569b37dd32852cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7121d0045745584a4a9bc0bd569b37dd32852cff"}]},"branch":"refs/heads/master"},"17b29764a89d915aabff50dfeca571a1ae719a85":{"kind":"REWORK","_number":17,"created":"2018-12-13 08:59:04.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/17"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-13 15:40:21.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests (WIP)\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/17b29764a89d915aabff50dfeca571a1ae719a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/17b29764a89d915aabff50dfeca571a1ae719a85"}]},"branch":"refs/heads/master"},"eda95ee692cbf48231f6ba8750053b31c8698fd8":{"kind":"REWORK","_number":18,"created":"2018-12-14 08:06:17.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/18"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-14 14:42:57.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests (WIP)\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eda95ee692cbf48231f6ba8750053b31c8698fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eda95ee692cbf48231f6ba8750053b31c8698fd8"}]},"branch":"refs/heads/master"},"8022dabb813bf32ac0a288488a169e25acb22b7c":{"kind":"REWORK","_number":19,"created":"2018-12-17 07:54:12.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/19"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-17 14:34:02.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests (WIP)\n- copy volume to image\n- copy image to volume\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8022dabb813bf32ac0a288488a169e25acb22b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8022dabb813bf32ac0a288488a169e25acb22b7c"}]},"branch":"refs/heads/master"},"33929701e83b413f32490adef519ef81b208851e":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2018-12-18 06:43:44.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/20"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-18 13:20:54.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- release-note\n- documentation\n\nTODO:\n- unit tests (WIP)\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I26e3dc140b2cf30a97665679fdc4d2f897cd4872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33929701e83b413f32490adef519ef81b208851e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33929701e83b413f32490adef519ef81b208851e"}]},"branch":"refs/heads/master"},"00feb1275993f70703bf85fcb9a247eb6847c71f":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2018-12-20 02:26:16.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/21"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-20 09:05:06.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nTODO:\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I26e3dc140b2cf30a97665679fdc4d2f897cd4872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/00feb1275993f70703bf85fcb9a247eb6847c71f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/00feb1275993f70703bf85fcb9a247eb6847c71f"}]},"branch":"refs/heads/master"},"5ac356c102ee20ef695ef57cfd237ad9c643b33c":{"kind":"REWORK","_number":22,"created":"2018-12-24 08:12:13.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/22"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-24 08:10:53.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nTODO:\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I26e3dc140b2cf30a97665679fdc4d2f897cd4872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ac356c102ee20ef695ef57cfd237ad9c643b33c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ac356c102ee20ef695ef57cfd237ad9c643b33c"}]},"branch":"refs/heads/master"},"8c0120b7f4c24181048cfb596dcbb7c22127e71d":{"kind":"REWORK","_number":23,"created":"2018-12-26 08:09:44.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/23"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-26 08:08:34.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nTODO:\n- rsd-lib volume extend\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I26e3dc140b2cf30a97665679fdc4d2f897cd4872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c0120b7f4c24181048cfb596dcbb7c22127e71d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c0120b7f4c24181048cfb596dcbb7c22127e71d"}]},"branch":"refs/heads/master"},"ac25d034bc47432721e2679d38b8df2bf37a78f6":{"kind":"REWORK","_number":24,"created":"2019-01-04 05:13:54.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/24"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-04 05:11:47.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nTODO:\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I26e3dc140b2cf30a97665679fdc4d2f897cd4872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac25d034bc47432721e2679d38b8df2bf37a78f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac25d034bc47432721e2679d38b8df2bf37a78f6"}]},"branch":"refs/heads/master"},"28d00871aa8ad12d4c123c644983c25071f31b62":{"kind":"REWORK","_number":25,"created":"2019-01-04 14:26:07.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/25"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-04 14:24:38.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support.\n\nDone:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nTODO:\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I26e3dc140b2cf30a97665679fdc4d2f897cd4872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28d00871aa8ad12d4c123c644983c25071f31b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28d00871aa8ad12d4c123c644983c25071f31b62"}]},"branch":"refs/heads/master"},"590dbe606d05bccc68e32d08db410edcb28c2dc9":{"kind":"REWORK","_number":26,"created":"2019-01-07 03:45:04.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/26"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-07 03:42:55.000000000","tz":480},"subject":"[wip] Cinder RSD Driver","message":"[wip] Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nTODO:\n- rsd-lib identify busy snapshot\n- verification\n- CI\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I26e3dc140b2cf30a97665679fdc4d2f897cd4872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/590dbe606d05bccc68e32d08db410edcb28c2dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/590dbe606d05bccc68e32d08db410edcb28c2dc9"}]},"branch":"refs/heads/master"},"9ed4b87c160f898fca10eafd0d0e7dcf26fb610f":{"kind":"REWORK","_number":27,"created":"2019-01-08 06:29:16.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/27"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-08 06:27:07.000000000","tz":480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I26e3dc140b2cf30a97665679fdc4d2f897cd4872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ed4b87c160f898fca10eafd0d0e7dcf26fb610f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ed4b87c160f898fca10eafd0d0e7dcf26fb610f"}]},"branch":"refs/heads/master"},"fdec8019ebb733ff4765ab9514cf8655c8ed3db2":{"kind":"REWORK","_number":28,"created":"2019-01-09 08:30:42.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/28"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-09 08:28:30.000000000","tz":480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I049f1d701c99612959d8262342aa63a043a97db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdec8019ebb733ff4765ab9514cf8655c8ed3db2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdec8019ebb733ff4765ab9514cf8655c8ed3db2"}]},"branch":"refs/heads/master"},"cfb2ca9fff182d680fb69f70c662cf258ceb8aa0":{"kind":"REWORK","_number":29,"created":"2019-01-10 21:13:51.000000000","uploader":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"ref":"refs/changes/65/621465/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/29"}}},"commit":{"parents":[{"commit":"da34b3557939d11aae2b03933641eb9143891db3","subject":"Merge \"Handle string status codes in logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da34b3557939d11aae2b03933641eb9143891db3"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","date":"2019-01-10 20:07:56.000000000","tz":0},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I049f1d701c99612959d8262342aa63a043a97db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cfb2ca9fff182d680fb69f70c662cf258ceb8aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cfb2ca9fff182d680fb69f70c662cf258ceb8aa0"}]},"branch":"refs/heads/master"},"efd0562f6855e15531341a07903b7b86463af634":{"kind":"REWORK","_number":30,"created":"2019-01-15 01:32:08.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/30"}}},"commit":{"parents":[{"commit":"27c7829d5075a7c96bacb858f83270e35104d8a1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27c7829d5075a7c96bacb858f83270e35104d8a1"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-15 01:30:37.000000000","tz":480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I049f1d701c99612959d8262342aa63a043a97db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efd0562f6855e15531341a07903b7b86463af634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efd0562f6855e15531341a07903b7b86463af634"}]},"branch":"refs/heads/master"},"d1d778ed75c569503485847ff4212ae8759be793":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2019-01-15 01:39:39.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/31"}}},"commit":{"parents":[{"commit":"39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd","subject":"Merge \"Implement volume capacity stats for VMware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-15 01:38:07.000000000","tz":480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I049f1d701c99612959d8262342aa63a043a97db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1d778ed75c569503485847ff4212ae8759be793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1d778ed75c569503485847ff4212ae8759be793"}]},"branch":"refs/heads/master"},"305a6c8993f65284e6ac79bd930f6b4be83dc7cf":{"kind":"REWORK","_number":32,"created":"2019-01-16 12:08:12.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/32"}}},"commit":{"parents":[{"commit":"39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd","subject":"Merge \"Implement volume capacity stats for VMware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-16 12:07:40.000000000","tz":480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I049f1d701c99612959d8262342aa63a043a97db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/305a6c8993f65284e6ac79bd930f6b4be83dc7cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/305a6c8993f65284e6ac79bd930f6b4be83dc7cf"}]},"branch":"refs/heads/master"},"604e1b099ae796b238c6dc1ab06492dc63c48367":{"kind":"REWORK","_number":33,"created":"2019-01-17 02:05:53.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/33"}}},"commit":{"parents":[{"commit":"39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd","subject":"Merge \"Implement volume capacity stats for VMware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-17 02:04:20.000000000","tz":480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I049f1d701c99612959d8262342aa63a043a97db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/604e1b099ae796b238c6dc1ab06492dc63c48367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/604e1b099ae796b238c6dc1ab06492dc63c48367"}]},"branch":"refs/heads/master"},"9718e813838eeb5f533024c0e779290495f1a8fd":{"kind":"REWORK","_number":34,"created":"2019-01-17 06:28:52.000000000","uploader":{"_account_id":15648,"name":"Yingxin","email":"yingxin.cheng@intel.com","username":"Yingxin"},"ref":"refs/changes/65/621465/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/34"}}},"commit":{"parents":[{"commit":"39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd","subject":"Merge \"Implement volume capacity stats for VMware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2019-01-17 06:28:16.000000000","tz":480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I049f1d701c99612959d8262342aa63a043a97db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9718e813838eeb5f533024c0e779290495f1a8fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9718e813838eeb5f533024c0e779290495f1a8fd"}]},"branch":"refs/heads/master"},"874f4ab23d55dcdd6e8e27cf0fee8d804b556166":{"kind":"REWORK","_number":35,"created":"2019-03-01 05:40:28.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/35"}}},"commit":{"parents":[{"commit":"39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd","subject":"Merge \"Implement volume capacity stats for VMware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-02-28 22:36:59.000000000","tz":-480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\nDepends-On: I049f1d701c99612959d8262342aa63a043a97db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/874f4ab23d55dcdd6e8e27cf0fee8d804b556166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/874f4ab23d55dcdd6e8e27cf0fee8d804b556166"}]},"branch":"refs/heads/master"},"be569f8861f8590a30838a208bb593a92d251f15":{"kind":"REWORK","_number":36,"created":"2019-03-08 23:03:59.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/36"}}},"commit":{"parents":[{"commit":"39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd","subject":"Merge \"Implement volume capacity stats for VMware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-03-08 14:44:01.000000000","tz":-480},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be569f8861f8590a30838a208bb593a92d251f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be569f8861f8590a30838a208bb593a92d251f15"}]},"branch":"refs/heads/master"},"48a29dc52d9fd88ed558f70ddb76a07b8454552c":{"kind":"REWORK","_number":37,"created":"2019-04-11 18:01:54.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/37"}}},"commit":{"parents":[{"commit":"39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd","subject":"Merge \"Implement volume capacity stats for VMware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39e78e44e0849bb62cea7f1bc60d1f7cd9c905dd"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"tkauthar","email":"tejdeep.kautharam@intel.com","date":"2019-04-09 16:59:27.000000000","tz":-420},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48a29dc52d9fd88ed558f70ddb76a07b8454552c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48a29dc52d9fd88ed558f70ddb76a07b8454552c"}]},"branch":"refs/heads/master"},"5c02fe8d23feafeb58ee94654da146e33b978dff":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2019-04-12 05:23:04.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/65/621465/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/38"}}},"commit":{"parents":[{"commit":"8c9a2fa27294b6a01aa4df9d7152f177cbebd456","subject":"Merge \"Document behavior of message.create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c9a2fa27294b6a01aa4df9d7152f177cbebd456"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2019-04-12 05:23:04.000000000","tz":0},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c02fe8d23feafeb58ee94654da146e33b978dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c02fe8d23feafeb58ee94654da146e33b978dff"}]},"branch":"refs/heads/master"},"b5912e16d4b8b443fe2a16c8c985e43abd73eece":{"kind":"REWORK","_number":39,"created":"2019-04-25 23:38:46.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/39"}}},"commit":{"parents":[{"commit":"8c9a2fa27294b6a01aa4df9d7152f177cbebd456","subject":"Merge \"Document behavior of message.create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c9a2fa27294b6a01aa4df9d7152f177cbebd456"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-04-25 23:18:43.000000000","tz":-420},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5912e16d4b8b443fe2a16c8c985e43abd73eece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5912e16d4b8b443fe2a16c8c985e43abd73eece"}]},"branch":"refs/heads/master"},"493559db4fe33481b6e97ce157437f5aafa0fbf5":{"kind":"REWORK","_number":40,"created":"2019-04-30 21:25:14.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/40"}}},"commit":{"parents":[{"commit":"8c9a2fa27294b6a01aa4df9d7152f177cbebd456","subject":"Merge \"Document behavior of message.create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c9a2fa27294b6a01aa4df9d7152f177cbebd456"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-04-30 21:23:29.000000000","tz":-420},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/493559db4fe33481b6e97ce157437f5aafa0fbf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/493559db4fe33481b6e97ce157437f5aafa0fbf5"}]},"branch":"refs/heads/master"},"4d150a10954c332b5d019a12d3fc72c893a06eaf":{"kind":"REWORK","_number":41,"created":"2019-05-03 07:33:33.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/41"}}},"commit":{"parents":[{"commit":"8c9a2fa27294b6a01aa4df9d7152f177cbebd456","subject":"Merge \"Document behavior of message.create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c9a2fa27294b6a01aa4df9d7152f177cbebd456"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-05-03 07:31:03.000000000","tz":-420},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d150a10954c332b5d019a12d3fc72c893a06eaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d150a10954c332b5d019a12d3fc72c893a06eaf"}]},"branch":"refs/heads/master"},"3f4a6d1bf3284a43c187b039cbb329fcfcd9fe2c":{"kind":"REWORK","_number":42,"created":"2019-05-03 18:34:44.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/42"}}},"commit":{"parents":[{"commit":"8c9a2fa27294b6a01aa4df9d7152f177cbebd456","subject":"Merge \"Document behavior of message.create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c9a2fa27294b6a01aa4df9d7152f177cbebd456"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-05-03 18:30:58.000000000","tz":-420},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f4a6d1bf3284a43c187b039cbb329fcfcd9fe2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f4a6d1bf3284a43c187b039cbb329fcfcd9fe2c"}]},"branch":"refs/heads/master"},"f5fe3251f5dcfffd3415bd0007e3ee3cb130ef44":{"kind":"REWORK","_number":43,"created":"2019-05-23 16:53:13.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/43"}}},"commit":{"parents":[{"commit":"8c9a2fa27294b6a01aa4df9d7152f177cbebd456","subject":"Merge \"Document behavior of message.create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c9a2fa27294b6a01aa4df9d7152f177cbebd456"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-05-07 23:35:43.000000000","tz":-420},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5fe3251f5dcfffd3415bd0007e3ee3cb130ef44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5fe3251f5dcfffd3415bd0007e3ee3cb130ef44"}]},"branch":"refs/heads/master"},"e67a9584aab5127daac0cddb20ef33c550c3c252":{"kind":"REWORK","_number":44,"created":"2019-05-28 21:41:09.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/44"}}},"commit":{"parents":[{"commit":"8c9a2fa27294b6a01aa4df9d7152f177cbebd456","subject":"Merge \"Document behavior of message.create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c9a2fa27294b6a01aa4df9d7152f177cbebd456"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-05-28 21:40:03.000000000","tz":-420},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e67a9584aab5127daac0cddb20ef33c550c3c252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e67a9584aab5127daac0cddb20ef33c550c3c252"}]},"branch":"refs/heads/master"},"b92b80241c86d74d71298077b3357989a9188b43":{"kind":"REWORK","_number":45,"created":"2019-06-06 21:54:04.000000000","uploader":{"_account_id":29964,"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","username":"tejdeep"},"ref":"refs/changes/65/621465/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/621465/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/621465/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/621465/45"}}},"commit":{"parents":[{"commit":"3c29c3846ee4e0a43e74f7328642ace218f00d18","subject":"Merge \"Replace assert to raise AssertionError\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c29c3846ee4e0a43e74f7328642ace218f00d18"}]}],"author":{"name":"Yingxin","email":"yingxin.cheng@intel.com","date":"2018-12-03 08:51:13.000000000","tz":480},"committer":{"name":"Tejdeep Kautharam","email":"tejdeep.kautharam@intel.com","date":"2019-06-06 21:43:04.000000000","tz":-420},"subject":"Cinder RSD Driver","message":"Cinder RSD Driver\n\nAdd a new Cinder Driver for Intel RSD with NVMeoF support:\n- report stats\n- create volume\n- delete volume\n- attach volume\n- detach volume\n- create snapshot\n- delete snapshot\n- create volume from snapshot\n- clone volume\n- extend volume\n- copy volume to image (reuse the base-class impl)\n- copy image to volume (reuse the base-class impl)\n- host-assisted volume migration (reuse the manager code)\n- unit tests\n- release-note\n- documentation\n\nChange-Id: Ie83dab4858729353865268231352033934dafdc2\nImplements: blueprint nvme-volume-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b92b80241c86d74d71298077b3357989a9188b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b92b80241c86d74d71298077b3357989a9188b43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
