)]}'
{"id":"openstack%2Fcinder~663518","triplet_id":"openstack%2Fcinder~stable%2Fqueens~Ib5947e2c7300730adb851ad58e898a29f2b88525","project":"openstack/cinder","branch":"stable/queens","topic":"bug/1801595","hashtags":[],"change_id":"Ib5947e2c7300730adb851ad58e898a29f2b88525","subject":"Create new image volume cache entry when cloning fails","status":"MERGED","created":"2019-06-06 06:46:11.000000000","updated":"2019-06-12 17:02:42.000000000","submitted":"2019-06-12 17:02:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663518-1560358962631-38ffb1f0","meta_rev_id":"711010b20003f17628489aad13758b7f7d8d22fb","_number":663518,"virtual_id_number":663518,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-06-06 22:13:06.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2019-06-06 13:01:11.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 17:34:57.000000000","_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 11:47:13.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"date":"2019-06-06 12:25:43.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":2,"date":"2019-06-12 17:02:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 16:10:43.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 21:39:27.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"date":"2019-06-07 01:14:09.000000000","_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 15:19:27.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2019-06-07 04:44:56.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-12 01:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":1,"date":"2019-06-12 00:32:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":2,"date":"2019-06-12 14:17:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":1,"date":"2019-06-12 14:17:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 00:42:02.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 11:47:13.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-06-06 12:25:43.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-06-06 13:01:11.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 15:19:27.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-06-06 16:10:43.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 17:34:57.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 22:13:06.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-06-07 01:14:09.000000000","updated_by":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 04:44:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 21:39:27.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2019-06-12 00:32:43.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2019-06-12 00:42:02.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-06-12 01:13:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-12 14:17:43.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-06-12 17:02:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"64084cd0a0adc2aad98e7578ce61dc200a473129","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-06-06 06:46:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db83cd652a1aaccb64fdf604548c73db146e633a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-06 07:33:58.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/663518/1/iscsi/ : FAILURE in 44m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/663518/1/smb/ : FAILURE in 47m 06s","accounts_in_message":[],"_revision_number":1},{"id":"879470db48de54642a4bb6385c4a40a4e23c13ac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-06 08:09:53.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/18/663518/1/check/tempest-dsvm-full-xenial/0becc02/ : SUCCESS in 1h 23m 05s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/18/663518/1/check/devstack-plugin-ceph-dsvm-xenial/85aa392/ : SUCCESS in 1h 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"797b331e8ca989e9b16575c008312d8b05ff888e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-06 08:26:51.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/18/663518/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b9f03200c1ac2d70bafaa206209034c874197aeb","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-06 09:15:51.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/18/663518/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/bc25ce7 : SUCCESS in 45m 02s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/18/663518/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/7fec6c1 : FAILURE in 1h 48m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"e54e5bab8bff2fdd326e9d1fc45e239c0ad4a3b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:41:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/18/663518/1/check/openstack-tox-pep8/acd6929/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/18/663518/1/check/openstack-tox-py27/cb84db0/ : FAILURE in 7m 49s\n- openstack-tox-py35 http://logs.openstack.org/18/663518/1/check/openstack-tox-py35/f6f1bc4/ : FAILURE in 9m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/663518/1/check/build-openstack-sphinx-docs/9f55de7/html/ : SUCCESS in 18m 37s\n- tempest-full http://logs.openstack.org/18/663518/1/check/tempest-full/df52532/ : SUCCESS in 1h 29m 55s\n- neutron-grenade http://logs.openstack.org/18/663518/1/check/neutron-grenade/8c6e77d/ : SUCCESS in 53m 13s\n- tempest-full-py3 http://logs.openstack.org/18/663518/1/check/tempest-full-py3/c770864/ : SUCCESS in 1h 16m 14s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/18/663518/1/check/cinder-tempest-dsvm-lvm-lio-barbican/d15786b/ : FAILURE in 1h 12m 43s (non-voting)\n- cinder-tox-bandit-baseline http://logs.openstack.org/18/663518/1/check/cinder-tox-bandit-baseline/be214da/ : TIMED_OUT in 40m 52s (non-voting)\n- nova-multiattach http://logs.openstack.org/18/663518/1/check/nova-multiattach/5066179/ : SUCCESS in 55m 08s\n- openstack-tox-pylint http://logs.openstack.org/18/663518/1/check/openstack-tox-pylint/672b88a/ : SUCCESS in 15m 43s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/18/663518/1/check/devstack-plugin-ceph-tempest/3093dff/ : SUCCESS in 1h 14m 09s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/18/663518/1/check/openstack-tox-functional/cced515/ : SUCCESS in 5m 07s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/18/663518/1/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/07d25f4/ : SUCCESS in 52m 15s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/18/663518/1/check/legacy-tempest-dsvm-lvm-multibackend/ba0d515/ : SUCCESS in 1h 07m 57s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/18/663518/1/check/legacy-tempest-dsvm-full-drbd-devstack/d6aec87/ : SUCCESS in 1h 09m 11s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/18/663518/1/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/22e1f40/ : SUCCESS in 1h 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3433440f64bd7dcdc9b292082d518afed1a3cab3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-06 11:25:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/18/663518/1/check/dsvm-tempest-cisco-zonemanager-job/b096034/ : FAILURE in 12m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b7c908ccfa33f2448470cf09d80657e423b29dd6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-06-06 11:46:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"035bd7cb9df9de922b97b1539debb07415cf0ade","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-06 11:47:13.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/18/663518/2/check/dsvm-tempest-infinibox-fc/a8ae927 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/18/663518/2/check/dsvm-tempest-infinibox-iscsi/ae8c27d : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"7a029ea3629db6b8326e558678fc72f64b9865b4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-06 11:49:34.000000000","message":"Patch Set 2: Code-Review-2\n\nNot merged to later branches yet.","accounts_in_message":[],"_revision_number":2},{"id":"187462b56001f2abbf04f98dfd05186d7845e676","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-06-06 12:25:43.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/663518/2/iscsi/ : FAILURE in 38m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/663518/2/smb/ : FAILURE in 38m 26s","accounts_in_message":[],"_revision_number":2},{"id":"07cb667e3250f2f0ffb3471dfbbd8c041b568e82","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-06 13:01:11.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/18/663518/2/check/tempest-dsvm-full-xenial/0e9f843/ : SUCCESS in 1h 12m 59s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/cinder/18/663518/2/check/devstack-plugin-ceph-dsvm-xenial/5aa7f33/ : SUCCESS in 1h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb6c1c8cad6811c932761b3b3dbb20b52eb0baae","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-06-06 15:19:27.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/18/663518/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"6b8138e0a64a7b0c9e83a5c8ed70736f657d18ae","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-06 16:10:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/18/663518/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/bdc1d19 : SUCCESS in 50m 10s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/18/663518/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/cb76efa : SUCCESS in 59m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"9cd4434656d863968762001f3fe3c2c40db65165","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 17:34:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/663518/2/check/openstack-tox-pep8/20cdfd9/ : SUCCESS in 7m 23s\n- openstack-tox-py27 http://logs.openstack.org/18/663518/2/check/openstack-tox-py27/389c261/ : SUCCESS in 10m 55s\n- openstack-tox-py35 http://logs.openstack.org/18/663518/2/check/openstack-tox-py35/6c4f4ff/ : SUCCESS in 7m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/663518/2/check/build-openstack-sphinx-docs/353798c/html/ : SUCCESS in 17m 27s\n- tempest-full http://logs.openstack.org/18/663518/2/check/tempest-full/4102eca/ : SUCCESS in 1h 24m 42s\n- neutron-grenade http://logs.openstack.org/18/663518/2/check/neutron-grenade/db97900/ : SUCCESS in 46m 07s\n- tempest-full-py3 http://logs.openstack.org/18/663518/2/check/tempest-full-py3/c01ea12/ : SUCCESS in 1h 24m 51s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/18/663518/2/check/cinder-tempest-dsvm-lvm-lio-barbican/4ec4b4b/ : FAILURE in 1h 12m 37s (non-voting)\n- cinder-tox-bandit-baseline http://logs.openstack.org/18/663518/2/check/cinder-tox-bandit-baseline/ee5c7c3/ : TIMED_OUT in 40m 52s (non-voting)\n- nova-multiattach http://logs.openstack.org/18/663518/2/check/nova-multiattach/888cb15/ : SUCCESS in 1h 04m 00s\n- openstack-tox-pylint http://logs.openstack.org/18/663518/2/check/openstack-tox-pylint/5d102f2/ : SUCCESS in 14m 52s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/18/663518/2/check/devstack-plugin-ceph-tempest/e64e41c/ : SUCCESS in 1h 16m 29s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/18/663518/2/check/openstack-tox-functional/d524fd8/ : SUCCESS in 5m 32s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/18/663518/2/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/2e72fbd/ : SUCCESS in 1h 18m 29s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/18/663518/2/check/legacy-tempest-dsvm-lvm-multibackend/a6590c8/ : SUCCESS in 1h 26m 13s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/18/663518/2/check/legacy-tempest-dsvm-full-drbd-devstack/4b57b9f/ : SUCCESS in 1h 16m 58s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/18/663518/2/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/3f813a7/ : SUCCESS in 1h 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6c54b6c2be71501068b4a33ea77059eb91c3e6f","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-06-06 17:34:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.179.158/logs/18/663518/2/upstream-eseries-check/cinder-eseries-iSCSI/2f53376/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://13.56.179.158/logs/18/663518/2/upstream-eseries-check/cinder-eseries-FCP/781ce6f/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":2},{"id":"f3646a99e506da5d43f84abe1348731a09a01e68","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-06-06 18:51:18.000000000","message":"Patch Set 2:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/18/2204 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"55bf18b3a00c06aab9775d4485df7381d7a9c97b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-06-06 22:13:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/18/663518/2/check/dsvm-tempest-cisco-zonemanager-job/ea393e7/ : FAILURE in 12m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5eca9028769a844f92ff9b9a4d1e58307348197c","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 01:14:09.000000000","message":"Patch Set 2:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-18-663518-2-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b2f3af6f6049d0d70735895c5f3af2e24609428e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 04:44:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/18/663518/2/check/NEC-ISCSI/23ae70a : SUCCESS in 1h 45m 30s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/18/663518/2/check/NEC-FC/54a93cb : SUCCESS in 1h 36m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6ef46d7c5641cfd45114df63c4d90b212b5e8f37","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-06-07 21:39:27.000000000","message":"Patch Set 2:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/18/2210 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"505eb179d33361aa1de595f6cd15855d3bfd7d0f","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-06-12 00:32:43.000000000","message":"Patch Set 2: Code-Review+1\n\nThe patch merged on rocky, and I agree it\u0027s suitable for queens.","accounts_in_message":[],"_revision_number":2},{"id":"6a07c6679cea40782764790517f852b44683ab96","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-12 01:13:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"faa0627c77f88ce06c9092c82980cec9cac7d6ee","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-06-12 14:17:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5766c2b9922a2ffe86c24661f568af423bead143","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 14:18:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"852f53408e3f24ad12a23999d41ef59c6763d23d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 17:02:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/663518/2/gate/openstack-tox-pep8/5f376c0/ : SUCCESS in 7m 35s\n- openstack-tox-py27 http://logs.openstack.org/18/663518/2/gate/openstack-tox-py27/fb10804/ : SUCCESS in 8m 30s\n- openstack-tox-py35 http://logs.openstack.org/18/663518/2/gate/openstack-tox-py35/0acb6b7/ : SUCCESS in 8m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/663518/2/gate/build-openstack-sphinx-docs/94eb080/html/ : SUCCESS in 16m 56s\n- tempest-full http://logs.openstack.org/18/663518/2/gate/tempest-full/ab57d1a/ : SUCCESS in 1h 23m 35s\n- neutron-grenade http://logs.openstack.org/18/663518/2/gate/neutron-grenade/2edf182/ : SUCCESS in 54m 35s\n- tempest-full-py3 http://logs.openstack.org/18/663518/2/gate/tempest-full-py3/6c09237/ : SUCCESS in 1h 25m 48s\n- nova-multiattach http://logs.openstack.org/18/663518/2/gate/nova-multiattach/f370eff/ : SUCCESS in 49m 29s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/18/663518/2/gate/legacy-grenade-dsvm-cinder-mn-sub-volbak/89e94b0/ : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":2},{"id":"031ac2119e7880aef5afa2d3bb7b5155d761eb5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 17:02:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1accbbc7a0912de9faffc35e1249bc4cf7d315ff","revisions":{"1e91d2d02ef6147685441fe83f6b6486fe497523":{"kind":"REWORK","_number":1,"created":"2019-06-06 06:46:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/663518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/663518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/663518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/663518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/663518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/663518/1"}}},"commit":{"parents":[{"commit":"8d7620b478e31510e5f1ec74eb8fb488d45a7873","subject":"Fix \"connector\u003dNone\" issue in Kaminario drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d7620b478e31510e5f1ec74eb8fb488d45a7873"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-01-14 17:42:40.000000000","tz":-300},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-06-06 06:45:40.000000000","tz":540},"subject":"Create new image volume cache entry when cloning fails","message":"Create new image volume cache entry when cloning fails\n\nThis patch fixes the image volume cache so that a new cache entry is\ncreated whenever cloning an existing entry fails (which can happen, for\nexample, when a cached volume reaches its snapshot limit). This restores\nthe original behavior, which was broken by [1].\n\n[1] I547fb4bcdd4783225b8ca96d157c61ca3bcf4ef4\n\nCloses-Bug: #1801595\nChange-Id: Ib5947e2c7300730adb851ad58e898a29f2b88525\n(cherry picked from commit 93519a02c0c9b5dcb9d5139761780bcbdceef982)\n(cherry picked from commit b0cf233d808507707f3ccf7c614d3930f2e3eb32)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e91d2d02ef6147685441fe83f6b6486fe497523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e91d2d02ef6147685441fe83f6b6486fe497523"}]},"branch":"refs/heads/stable/queens"},"1accbbc7a0912de9faffc35e1249bc4cf7d315ff":{"kind":"REWORK","_number":2,"created":"2019-06-06 11:46:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/663518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/663518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/663518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/663518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/663518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/663518/2"}}},"commit":{"parents":[{"commit":"8d7620b478e31510e5f1ec74eb8fb488d45a7873","subject":"Fix \"connector\u003dNone\" issue in Kaminario drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d7620b478e31510e5f1ec74eb8fb488d45a7873"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-01-14 17:42:40.000000000","tz":-300},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-06-06 11:42:08.000000000","tz":0},"subject":"Create new image volume cache entry when cloning fails","message":"Create new image volume cache entry when cloning fails\n\nThis patch fixes the image volume cache so that a new cache entry is\ncreated whenever cloning an existing entry fails (which can happen, for\nexample, when a cached volume reaches its snapshot limit). This restores\nthe original behavior, which was broken by [1].\n\n[1] I547fb4bcdd4783225b8ca96d157c61ca3bcf4ef4\n\nConflicts:\n        cinder/tests/unit/volume/flows/test_create_volume_flow.py\n\nCloses-Bug: #1801595\nChange-Id: Ib5947e2c7300730adb851ad58e898a29f2b88525\n(cherry picked from commit 93519a02c0c9b5dcb9d5139761780bcbdceef982)\n(cherry picked from commit b0cf233d808507707f3ccf7c614d3930f2e3eb32)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1accbbc7a0912de9faffc35e1249bc4cf7d315ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1accbbc7a0912de9faffc35e1249bc4cf7d315ff"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
