)]}'
{"id":"openstack%2Fcinder~670715","triplet_id":"openstack%2Fcinder~master~I11df21a7a213a0700d277d57ceee40f636955c50","project":"openstack/cinder","branch":"master","topic":"zadara_api_access_key_auth","hashtags":[],"change_id":"I11df21a7a213a0700d277d57ceee40f636955c50","subject":"Zadara VPSA: Move to API access key authentication","status":"MERGED","created":"2019-07-14 08:15:48.000000000","updated":"2019-09-25 19:39:37.000000000","submitted":"2019-09-25 19:38:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670715-1569440282878-b42f53c0","meta_rev_id":"0a21228a73d70c521e4041045be281379f6debde","_number":670715,"virtual_id_number":670715,"owner":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-09-11 17:50:33.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 15:19:04.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2019-09-12 23:38:52.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 18:37:53.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 00:28:52.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 07:33:36.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 03:42:55.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 19:38:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 22:16:31.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2019-09-11 18:50:11.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-13 06:58:50.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-09-13 18:37:02.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"date":"2019-09-12 14:55:17.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 09:12:00.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2019-09-12 05:44:58.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 02:59:48.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 02:58:37.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 01:41:00.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"date":"2019-09-12 17:56:05.000000000","_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"date":"2019-09-11 09:57:21.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-09-12 04:25:59.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 17:10:16.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"date":"2019-09-11 15:17:19.000000000","_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 01:43:15.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 20:21:06.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2019-09-12 21:18:09.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2019-09-11 20:39:52.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"date":"2019-09-12 01:53:18.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 15:58:45.000000000","_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-12 21:52:29.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"date":"2019-09-11 10:00:18.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 07:42:37.000000000","_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-19 14:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":2,"date":"2019-09-25 17:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":1,"date":"2019-09-25 17:11:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 13:26:03.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2019-08-06 14:11:40.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-25 10:14:39.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 09:41:58.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-09-03 16:40:34.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-09-05 09:19:17.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-07 20:13:33.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-07 20:23:24.000000000","updated_by":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-07 22:55:41.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-10 19:24:32.000000000","updated_by":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"reviewer":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-09-11 07:33:36.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-11 07:42:37.000000000","updated_by":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"reviewer":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-11 09:12:00.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-09-11 09:57:21.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-09-11 10:00:18.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-11 15:17:19.000000000","updated_by":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"reviewer":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"state":"REVIEWER"},{"updated":"2019-09-11 15:19:04.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-09-11 15:58:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-11 17:10:16.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-09-11 17:50:33.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-11 18:50:11.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-11 20:39:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 00:28:52.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 01:41:00.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-09-12 01:43:15.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 01:53:18.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 02:58:37.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 02:59:48.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 03:42:55.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 04:25:59.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 05:44:58.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 14:55:17.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-09-12 17:56:05.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2019-09-12 18:37:53.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 20:21:06.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2019-09-12 21:18:09.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2019-09-12 21:52:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-09-12 22:16:31.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-09-12 23:38:52.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-13 06:58:50.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2019-09-13 18:37:02.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2019-09-19 14:27:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-25 17:11:20.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-09-25 19:38:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21acf8a27f197e5b8a346f15954484ac7e3d05ee","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-07-14 08:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a5b563ec3c023475085eddd9cae9ddfd2f35a64","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-14 08:17:05.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3519463bfe1a13a231d349c48aa327aadf39e643","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-14 08:22:43.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1735 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1685 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1887 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"aab183b0aef84b5a5cea4c1bc27053c9f11349b6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-14 08:34:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/1/check/dsvm-tempest-zadara-driver/9664b99 : ABORTED in 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39e9c6180fd7bbed956343499bd479ede66ba61a","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-07-14 08:35:30.000000000","message":"Patch Set 1:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":1},{"id":"e851a7258fd0e027f9a51cd43feb3650a05c04c8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-14 08:41:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/1/check/dsvm-tempest-zadara-driver/61708b2 : FAILURE in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6230f1838b783fadce5c61bf32188ab66c87e647","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-14 08:55:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/1/check/dsvm-tempest-kvm/a4a3da2 : FAILURE in 34m 06s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/1/check/dsvm-tempest-exe/1965985 : FAILURE in 34m 55s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"b543a1d59ed80c678da1ca1f45596b8c5845122e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-14 08:59:46.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/1/check-not-vote/ext-cinder-zuul/bdf0717 : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ea0251701f2e759233e7e5a6ceb56003f00b07d9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-14 09:08:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/1/iscsi/ : SUCCESS in 51m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/1/smb/ : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8467623cf8a1311d6b5ecd9ab440d9344d88296a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-14 09:29:48.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/1/check/DellEMC_Unity_FC/d07cc66/DellEMC_Unity_FC/661 : SUCCESS in 1h 05m 18s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/1/check/DellEMC_Unity_iSCSI/1779075/DellEMC_Unity_iSCSI/630 : SUCCESS in 1h 13m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"637629ef06e4f66f29fcab30ca3482d797c06864","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-14 09:37:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/1/check/dsvm-tempest-cisco-zonemanager-job/af0852f/ : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7bb930b9f3bef737e46874c0f1b9f7ab938076cd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-14 09:45:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/1/check/tempest-dsvm-full-xenial/37ecfa0/ : SUCCESS in 1h 22m 41s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/1/check/devstack-plugin-ceph-dsvm-xenial/3e1a683/ : SUCCESS in 1h 29m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0023563f7b2c8febd26ab9b5ad9641ca45e6b3f4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-14 09:46:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/1/check/dsvm-tempest-infinibox-fc/cb82e4d : SUCCESS in 1h 12m 31s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/1/check/dsvm-tempest-infinibox-iscsi/77c05f5 : SUCCESS in 1h 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d1d7c98c1de3a8c7a6eee0463362e7cd00b88c2c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-14 09:50:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/1/check/DellEMC_XtremIO_FC/6f2dd75/DellEMC_XtremIO_FC/838 : SUCCESS in 1h 31m 43s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/1/check/DellEMC_XtremIO_iSCSI/73c342e/DellEMC_XtremIO_iSCSI/785 : SUCCESS in 1h 34m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"c1fde41022e71d6dda40b1de39e553f170d20820","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-14 10:40:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/1/check/NEC-ISCSI/31120c0 : SUCCESS in 2h 17m 49s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/1/check/NEC-FC/a7ea67f : SUCCESS in 2h 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d99553890326d2abb49baf7d6a5e12233862f0ff","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-07-14 10:44:26.000000000","message":"Patch Set 1:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":1},{"id":"10e2e39c2316942dd4ad4afc787822eac2ea9574","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-14 10:45:20.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/15/670715/1/check-cinder/Cinder-ISER-ISCSI/61e2db9 : FAILURE in 1h 07m 22s\n- Cinder-ISER-LIO http://13.74.249.42/15/670715/1/check-cinder/Cinder-ISER-LIO/0508b11 : SUCCESS in 1h 21m 12s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"da08146dd83c9419cf73a8f7068ccedfc0398226","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 10:59:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/670715/1/check/openstack-tox-lower-constraints/8d5e035/ : SUCCESS in 12m 10s\n- openstack-tox-pep8 http://logs.openstack.org/15/670715/1/check/openstack-tox-pep8/8d15652/ : SUCCESS in 9m 30s\n- openstack-tox-py27 http://logs.openstack.org/15/670715/1/check/openstack-tox-py27/186ea90/ : SUCCESS in 8m 57s\n- openstack-tox-py36 http://logs.openstack.org/15/670715/1/check/openstack-tox-py36/e1828cc/ : SUCCESS in 10m 33s\n- openstack-tox-py37 http://logs.openstack.org/15/670715/1/check/openstack-tox-py37/0af9862/ : SUCCESS in 11m 27s\n- openstack-tox-docs http://logs.openstack.org/15/670715/1/check/openstack-tox-docs/6fbf378/html/ : SUCCESS in 6m 41s\n- tempest-full http://logs.openstack.org/15/670715/1/check/tempest-full/213567c/ : SUCCESS in 1h 27m 09s\n- neutron-grenade http://logs.openstack.org/15/670715/1/check/neutron-grenade/5c89c31/ : SUCCESS in 55m 35s\n- grenade-py3 http://logs.openstack.org/15/670715/1/check/grenade-py3/cec0193/ : SUCCESS in 50m 52s\n- tempest-full-py3 http://logs.openstack.org/15/670715/1/check/tempest-full-py3/7e9227a/ : SUCCESS in 1h 09m 34s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/15/670715/1/check/cinder-tempest-dsvm-lvm-lio-barbican/ecc36f3/ : FAILURE in 1h 18m 43s\n- cinder-tox-bandit-baseline http://logs.openstack.org/15/670715/1/check/cinder-tox-bandit-baseline/9edbe08/ : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/15/670715/1/check/openstack-tox-functional-py36/6d8fb36/ : SUCCESS in 5m 15s\n- cinder-rally-task http://logs.openstack.org/15/670715/1/check/cinder-rally-task/caba856/ : SUCCESS in 36m 13s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/15/670715/1/check/openstack-tox-pylint/87ad4bb/ : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/15/670715/1/check/cinder-plugin-ceph-tempest/724e294/ : SUCCESS in 1h 07m 24s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/15/670715/1/check/openstack-tox-functional/9032e71/ : SUCCESS in 4m 27s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/15/670715/1/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/74d4340/ : SUCCESS in 56m 53s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/15/670715/1/check/legacy-tempest-dsvm-lvm-multibackend/c17d525/ : SUCCESS in 1h 28m 18s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/15/670715/1/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/b614ff2/ : SUCCESS in 1h 36m 07s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/15/670715/1/check/tempest-slow-py3/681aacd/ : SUCCESS in 2h 41m 49s","accounts_in_message":[],"_revision_number":1},{"id":"23c4c39c02fefb6a8a35c3252011e70297a1fbbf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-14 11:40:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77708badd667e85b06c5562b4abc5a181e50828e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-14 12:28:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/1/check/dsvm-tempest-zadara-driver/09a1999 : SUCCESS in 1h 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0632f9404111a7a3565bd9cd1e963171c8619c98","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-14 15:39:59.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/15/670715/1/check/DellEMC_PowerMAX_FC/a5de607/DellEMC_PowerMAX_FC/425 : SUCCESS in 5h 18m 03s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/15/670715/1/check/DellEMC_PowerMAX_iSCSI/97a4460/DellEMC_PowerMAX_iSCSI/411 : SUCCESS in 7h 23m 56s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"4e9f4915b37dfdca4d7e2ca737540b4ebb6b0c49","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-07-15 07:11:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"68f55f1cd9353d1972e35ddf202c1ea7e25e0b07","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-15 07:18:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1737 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1689 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1889 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f272163018c81b11f61d28c074396001e7b42b9d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-15 08:25:57.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/2/check-not-vote/ext-cinder-zuul/4c0ad3b : FAILURE in 1h 13m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b6b4b997cf515013b8042502c7220ad62c58d2bc","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 08:32:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/2/check/dsvm-tempest-kvm/f847684 : FAILURE in 34m 30s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/2/check/dsvm-tempest-exe/f35a9c5 : FAILURE in 1h 12m 40s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"97879ef97096d5f5e88b0fe810fb1a77ce2a9346","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-15 08:33:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/2/check/dsvm-tempest-infinibox-fc/8d4349a : SUCCESS in 55m 52s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/2/check/dsvm-tempest-infinibox-iscsi/fe3f10d : SUCCESS in 1h 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8dddb46a0cc51e72bd144a261a6819de47c44f0f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-15 08:33:47.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/2/iscsi/ : FAILURE in 1h 21m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/2/smb/ : FAILURE in 50m 57s","accounts_in_message":[],"_revision_number":2},{"id":"0e7fa6e51cfda43cc02a41180994c11f6e130b48","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-15 08:34:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/2/check/tempest-dsvm-full-xenial/a67f1ee/ : SUCCESS in 1h 22m 46s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/2/check/devstack-plugin-ceph-dsvm-xenial/42062b2/ : SUCCESS in 1h 22m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df0e87cfd999b7d19a49540e2acb694c3f93268c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-15 08:49:34.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/2/check/fujitsu-eternus-dx-iscsi/d08eb3a : SUCCESS in 1h 35m 26s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/2/check/fujitsu-eternus-dx-fc/7eda98a : SUCCESS in 1h 37m 00s","accounts_in_message":[],"_revision_number":2},{"id":"555fcf2cac1211d2e321555aee69620896e317c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 09:48:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/670715/2/check/openstack-tox-lower-constraints/e5c876d/ : SUCCESS in 13m 04s\n- openstack-tox-pep8 http://logs.openstack.org/15/670715/2/check/openstack-tox-pep8/6ae6e29/ : SUCCESS in 8m 15s\n- openstack-tox-py27 http://logs.openstack.org/15/670715/2/check/openstack-tox-py27/cb88dbd/ : SUCCESS in 14m 13s\n- openstack-tox-py36 http://logs.openstack.org/15/670715/2/check/openstack-tox-py36/a533858/ : SUCCESS in 13m 11s\n- openstack-tox-py37 http://logs.openstack.org/15/670715/2/check/openstack-tox-py37/ea1682e/ : SUCCESS in 15m 05s\n- openstack-tox-docs http://logs.openstack.org/15/670715/2/check/openstack-tox-docs/ec5ead8/html/ : SUCCESS in 7m 59s\n- tempest-full http://logs.openstack.org/15/670715/2/check/tempest-full/8e9c36e/ : TIMED_OUT in 2h 13m 08s\n- neutron-grenade http://logs.openstack.org/15/670715/2/check/neutron-grenade/555f95d/ : SUCCESS in 1h 38m 18s\n- grenade-py3 http://logs.openstack.org/15/670715/2/check/grenade-py3/54f8449/ : SUCCESS in 1h 00m 14s\n- tempest-full-py3 http://logs.openstack.org/15/670715/2/check/tempest-full-py3/b0cd144/ : SUCCESS in 1h 24m 10s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/15/670715/2/check/cinder-tempest-dsvm-lvm-lio-barbican/3362b14/ : SUCCESS in 1h 25m 43s\n- cinder-tox-bandit-baseline http://logs.openstack.org/15/670715/2/check/cinder-tox-bandit-baseline/50d0ced/ : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/15/670715/2/check/openstack-tox-functional-py36/eb13b8a/ : SUCCESS in 5m 45s\n- cinder-rally-task http://logs.openstack.org/15/670715/2/check/cinder-rally-task/e0510ab/ : SUCCESS in 40m 54s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/15/670715/2/check/openstack-tox-pylint/555694b/ : SUCCESS in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/15/670715/2/check/cinder-plugin-ceph-tempest/af9800d/ : SUCCESS in 1h 34m 18s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/15/670715/2/check/openstack-tox-functional/b9e0569/ : SUCCESS in 6m 26s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/15/670715/2/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/7280da1/ : SUCCESS in 1h 09m 57s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/15/670715/2/check/legacy-tempest-dsvm-lvm-multibackend/f306b95/ : SUCCESS in 1h 27m 41s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/15/670715/2/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2dcdaac/ : SUCCESS in 1h 24m 20s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/15/670715/2/check/tempest-slow-py3/f23bb5c/ : SUCCESS in 2h 25m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3854173f04f545b22de085b0784a3511971d4170","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-15 09:55:59.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/2/check/DellEMC_Unity_FC/167c6bc/DellEMC_Unity_FC/665 : SUCCESS in 1h 11m 05s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/2/check/DellEMC_Unity_iSCSI/5f81063/DellEMC_Unity_iSCSI/634 : SUCCESS in 1h 20m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"704c1484d50b9175c2cac98a360064c94740a0b2","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-07-15 10:00:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"357a27be58501686def13b41857cfaf8d796c9e9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-15 10:25:27.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1a2959aa72dec2bae754855bbc295da456d4d865","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 10:28:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/2/check/DellEMC_XtremIO_FC/5a1c6f3/DellEMC_XtremIO_FC/842 : SUCCESS in 1h 27m 55s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/2/check/DellEMC_XtremIO_iSCSI/bd1837e/DellEMC_XtremIO_iSCSI/789 : SUCCESS in 1h 28m 16s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"f54933ef4286f984f2e585785dac5e2b31e0b878","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-15 10:55:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/2/iscsi/ : SUCCESS in 54m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/2/smb/ : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":2},{"id":"db790610458a594791f084fda73b2e1d982f4400","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-15 10:58:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/2/check/dsvm-tempest-cisco-zonemanager-job/63db9e7/ : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":2},{"id":"052f0c2bad248db349eed6526a0c632dba4b7639","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-15 11:02:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/2/check/dsvm-tempest-zadara-driver/7f602ce : SUCCESS in 1h 37m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d91753bde01bbc163fc05605142156f510fe5b1c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 11:13:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/2/check/dsvm-tempest-kvm/a548722 : FAILURE in 33m 36s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/2/check/dsvm-tempest-exe/74d1e1b : FAILURE in 41m 53s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"850336a55602280abba81877f472ea82b73e76e8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-15 11:36:40.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/2/check/tempest-dsvm-full-xenial/4a13c75/ : SUCCESS in 1h 23m 41s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/2/check/devstack-plugin-ceph-dsvm-xenial/a5c0c18/ : SUCCESS in 1h 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aaad5ab94d2ab4898153eacba3186fee5dacecb2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-15 11:59:47.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/2/check-not-vote/ext-cinder-zuul/cbffb9d : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":2},{"id":"be54f346357917a034c0d72599f5211f8d6483ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 13:05:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/670715/2/check/openstack-tox-lower-constraints/ca16db2/ : SUCCESS in 10m 20s\n- openstack-tox-pep8 http://logs.openstack.org/15/670715/2/check/openstack-tox-pep8/87c100a/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/15/670715/2/check/openstack-tox-py27/c2d01fb/ : SUCCESS in 9m 07s\n- openstack-tox-py36 http://logs.openstack.org/15/670715/2/check/openstack-tox-py36/6aef8eb/ : SUCCESS in 7m 52s\n- openstack-tox-py37 http://logs.openstack.org/15/670715/2/check/openstack-tox-py37/9d1fc3c/ : SUCCESS in 9m 07s\n- openstack-tox-docs http://logs.openstack.org/15/670715/2/check/openstack-tox-docs/0ec70d0/html/ : SUCCESS in 7m 12s\n- tempest-full http://logs.openstack.org/15/670715/2/check/tempest-full/193aa60/ : SUCCESS in 1h 37m 35s\n- neutron-grenade http://logs.openstack.org/15/670715/2/check/neutron-grenade/0121c66/ : SUCCESS in 57m 35s\n- grenade-py3 http://logs.openstack.org/15/670715/2/check/grenade-py3/0017211/ : SUCCESS in 59m 38s\n- tempest-full-py3 http://logs.openstack.org/15/670715/2/check/tempest-full-py3/4f461ad/ : SUCCESS in 1h 25m 22s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/15/670715/2/check/cinder-tempest-dsvm-lvm-lio-barbican/7a764a1/ : SUCCESS in 1h 39m 39s\n- cinder-tox-bandit-baseline http://logs.openstack.org/15/670715/2/check/cinder-tox-bandit-baseline/c70ce38/ : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/15/670715/2/check/openstack-tox-functional-py36/c272b79/ : SUCCESS in 8m 18s\n- cinder-rally-task http://logs.openstack.org/15/670715/2/check/cinder-rally-task/4d7bf6c/ : SUCCESS in 38m 57s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/15/670715/2/check/openstack-tox-pylint/9e19ad0/ : SUCCESS in 4m 48s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/15/670715/2/check/cinder-plugin-ceph-tempest/ea33457/ : SUCCESS in 1h 16m 26s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/15/670715/2/check/openstack-tox-functional/5725d1b/ : SUCCESS in 4m 56s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/15/670715/2/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/5e07ea3/ : SUCCESS in 1h 06m 41s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/15/670715/2/check/legacy-tempest-dsvm-lvm-multibackend/366f056/ : SUCCESS in 1h 17m 57s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/15/670715/2/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/897ba15/ : SUCCESS in 1h 35m 17s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/15/670715/2/check/tempest-slow-py3/2ae8fab/ : SUCCESS in 2h 44m 13s","accounts_in_message":[],"_revision_number":2},{"id":"2e584f4bea6320a72b1823219d7ae98ce2aeeeef","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-15 13:42:18.000000000","message":"Patch Set 2:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-15-670715-2/ns5_iscsi : FAILURE in 32m 56s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-15-670715-2/ns5_nfs : FAILURE in 32m 28s","accounts_in_message":[],"_revision_number":2},{"id":"24e4bbe9e59967e09768b26aa4a7f596626aa540","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-15 14:04:50.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-670715-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"245345be00805b6a465e04d84a5564f291abdd4b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-07-15 14:20:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/15/670715/2/upstream-solidfire-check/cinder-solidfire-iSCSI/b7179a6/ : SUCCESS in 1h 14m 27s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"ea8a477376191c3e4acb177af13f095498f969ff","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-15 14:49:17.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e14b57e : FAILURE in 1h 41m 23s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/2/check/3par-fc-driver-master-client-pip-ssa02-dsvm/b99610c : FAILURE in 1h 29m 35s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/15/670715/2/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/3d75f30 : FAILURE in 1h 22m 28s","accounts_in_message":[],"_revision_number":2},{"id":"6ac6de469927506db223b7c3ad11e54c194454c9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-15 15:01:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/15/670715/2/upstream-check/cinder-cDOT-iSCSI/9bd7542/ : SUCCESS in 57m 11s\n- cinder-cDOT-NFS http://13.56.179.158/logs/15/670715/2/upstream-check/cinder-cDOT-NFS/1e04137/ : SUCCESS in 59m 01s\n- cinder-cDOT-FCP http://13.56.179.158/logs/15/670715/2/upstream-check/cinder-cDOT-FCP/1f81501/ : FAILURE in 1h 55m 45s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2eea830706feb9973e3019da2060c1814d80aa99","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 15:12:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/2/check/DellEMC_XtremIO_FC/1103564/DellEMC_XtremIO_FC/845 : SUCCESS in 1h 27m 25s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/2/check/DellEMC_XtremIO_iSCSI/b9072a8/DellEMC_XtremIO_iSCSI/792 : SUCCESS in 1h 37m 32s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"d30c4feefcbb0e3dda1088aa0520e74cdd2c2730","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 15:20:19.000000000","message":"Patch Set 2:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-15-670715-2-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"539c433d4ddf4f52d2f554a6d274ce0cf4688677","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-15 15:31:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/2/check/DellEMC_Unity_FC/c594acf/DellEMC_Unity_FC/669 : SUCCESS in 1h 09m 19s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/2/check/DellEMC_Unity_iSCSI/14bd16b/DellEMC_Unity_iSCSI/638 : SUCCESS in 1h 26m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1c3ccce2a84137ddd3ca4976147010d5ba396e20","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-07-15 15:39:37.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/15/670715/2/check/kaminario-dsvm-tempest-full-iscsi/5ee4be9 : SUCCESS in 1h 49m 09s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/15/670715/2/check/kaminario-dsvm-tempest-full-FC/6dea8ba : FAILURE in 33m 51s","accounts_in_message":[],"_revision_number":2},{"id":"499d518201328ecd87441b2eaf97f960348caf38","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-07-15 15:47:40.000000000","message":"Patch Set 2:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/15/2359 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7fb3dd10304ede19c61a4e8be97162ab8a6a3188","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-07-15 16:41:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/670715/2/check/DellEMC_SC_FC/1adba1a/DellEMC_SC_FC/274 : SUCCESS in 59m 42s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/670715/2/check/DellEMC_SC_iSCSI/8562357/DellEMC_SC_iSCSI/257 : SUCCESS in 1h 35m 50s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3205ab0977141331f9080fcdaff5b972b58326a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 16:59:05.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/15/670715/2/check/DellEMC_VxFlexOS/8303ae5/DellEMC_VxFlexOS/435 : SUCCESS in 1h 48m 11s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"d3107a5d6d9b4ea0902e895da88f16fbf576f775","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-07-15 17:15:21.000000000","message":"Patch Set 2:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/15/670715/2/HPMSA-FC : SUCCESS in 44:09.53\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/15/670715/2/HPMSA-iSCSI : SUCCESS in 44:08.33","accounts_in_message":[],"_revision_number":2},{"id":"69182fdffe11fa99e5a13f4d533168dbd72e6025","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 21:26:03.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/15/670715/2/check/DellEMC_PowerMAX_FC/2d565c9/DellEMC_PowerMAX_FC/429 : FAILURE in 6h 12m 53s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/15/670715/2/check/DellEMC_PowerMAX_iSCSI/b4086e3/DellEMC_PowerMAX_iSCSI/415 : FAILURE in 7h 49m 13s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"8cec309872b0d1a57fae77923939458b80d2cb78","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-07-16 09:02:25.000000000","message":"Patch Set 2:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/15/670715/2/Lenovo-FC : SUCCESS in 38:21.87\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/15/670715/2/Lenovo-iSCSI : SUCCESS in 36:53.79","accounts_in_message":[],"_revision_number":2},{"id":"3bdbaa633f93d9a7cd7957de91e9b6349c5cd535","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-16 11:29:10.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1689 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1889 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1756 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e6be5de2b37d39c6a2aef19785f2242fbccb9a12","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-07-16 13:42:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1735 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1887 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1708 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b06288cff3b2b178d24d71461ea4c6d8bdd19430","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-19 03:04:50.000000000","message":"Patch Set 2: Code-Review+2\n\nMy comments have been addressed and CI reporting fine. LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"ee1e5a2a4bcadd90456217707bb4b71adc95eaaf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-19 19:33:47.000000000","message":"Patch Set 2: Code-Review-1\n\nOld config options need to be deprecated and kept around for one release before they are removed. Also needs a release note notifying admins that there is a change here that they should be aware of before upgrading. Could also include an upgrade check, but I would not block on that.","accounts_in_message":[],"_revision_number":2},{"id":"de6cb28a74f2b190a9cb5b0dc9599dc6372230eb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-19 19:36:32.000000000","message":"Patch Set 2:\n\nhttps://docs.openstack.org/cinder/latest/configuration/block-storage/drivers/zadara-volume-driver.html also needs to be updated.","accounts_in_message":[],"_revision_number":2},{"id":"708533094cc1c792dd0475e9ffaa06966a3032eb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-20 04:49:30.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"31b415c1a1228ae0a1469cabc350d3a707086e89","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-07-23 14:43:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c6d89237594ed606d79ecd734d4f09da91a1d59c","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-23 15:04:37.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1832 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1790 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1984 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"dba75dde871d2aa80d3240aae7993775ca7102a5","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-07-23 15:06:56.000000000","message":"Patch Set 3:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":3},{"id":"3030f71647cc5c52c206665f78650c6f0068a421","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-23 15:38:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/3/iscsi/ : SUCCESS in 54m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/3/smb/ : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":3},{"id":"2f5f548155c7441612746abff1309727b43e7580","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-23 15:38:40.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/3/check/dsvm-tempest-infinibox-fc/a30e065 : SUCCESS in 54m 27s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/3/check/dsvm-tempest-infinibox-iscsi/8e703b2 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"fc524ed40adfe1562f06d3d77f2b87b473390b84","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-23 15:41:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/3/check/dsvm-tempest-kvm/544a470 : FAILURE in 35m 51s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/3/check/dsvm-tempest-exe/118dd07 : FAILURE in 35m 24s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"c7e3f6eacb29ebbf5ad0c6d8c46d30b7e5645b7d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-23 15:48:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/a161723 : SUCCESS in 51m 52s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e2bd631 : SUCCESS in 1h 04m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"0014194d3582a3c3fe139543229be0b8976cdac2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-23 16:20:40.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/3/check/tempest-dsvm-full-xenial/09c5003/ : SUCCESS in 1h 33m 21s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/3/check/devstack-plugin-ceph-dsvm-xenial/7367855/ : SUCCESS in 1h 36m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bc567caf0dbd31e4cefa95b27283c1c0b378fcd8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-23 16:25:15.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"0c93f4a6f8e8055ad0f4dac802d0be7510acc69a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-23 16:36:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/3/check/dsvm-tempest-cisco-zonemanager-job/0d9169f/ : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":3},{"id":"d7bb5633c54837c0c48468275e7313c19bd6b47f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-23 16:52:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/3/check/dsvm-tempest-zadara-driver/297d381 : SUCCESS in 1h 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"decb799b89f3f170ebfc2c1a6edcf9a9fbbbbb32","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-23 18:15:19.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/3/check/NEC-ISCSI/0b61dd0 : FAILURE in 27m 14s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/3/check/NEC-FC/096a486 : SUCCESS in 2h 53m 34s","accounts_in_message":[],"_revision_number":3},{"id":"726b0263de0eb19163a4a21364858d2f724c2265","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 19:12:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/670715/3/check/openstack-tox-lower-constraints/d21ac26/ : SUCCESS in 12m 42s\n- openstack-tox-pep8 http://logs.openstack.org/15/670715/3/check/openstack-tox-pep8/e2ff908/ : SUCCESS in 7m 39s\n- openstack-tox-py27 http://logs.openstack.org/15/670715/3/check/openstack-tox-py27/04871a2/ : FAILURE in 6m 17s\n- openstack-tox-py36 http://logs.openstack.org/15/670715/3/check/openstack-tox-py36/57a08ba/ : SUCCESS in 9m 55s\n- openstack-tox-py37 http://logs.openstack.org/15/670715/3/check/openstack-tox-py37/9e99a2e/ : SUCCESS in 9m 56s\n- openstack-tox-docs http://logs.openstack.org/15/670715/3/check/openstack-tox-docs/607a491/ : FAILURE in 5m 59s\n- tempest-full http://logs.openstack.org/15/670715/3/check/tempest-full/aebd992/ : SUCCESS in 1h 36m 19s\n- neutron-grenade http://logs.openstack.org/15/670715/3/check/neutron-grenade/834671f/ : SUCCESS in 1h 02m 11s\n- grenade-py3 http://logs.openstack.org/15/670715/3/check/grenade-py3/c535602/ : SUCCESS in 1h 02m 05s\n- tempest-full-py3 http://logs.openstack.org/15/670715/3/check/tempest-full-py3/9b5fb60/ : SUCCESS in 1h 31m 57s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/15/670715/3/check/cinder-tempest-dsvm-lvm-lio-barbican/3db5feb/ : SUCCESS in 1h 30m 22s\n- cinder-tox-bandit-baseline http://logs.openstack.org/15/670715/3/check/cinder-tox-bandit-baseline/1a53008/ : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/15/670715/3/check/openstack-tox-functional-py36/fba81ca/ : SUCCESS in 7m 00s\n- cinder-rally-task http://logs.openstack.org/15/670715/3/check/cinder-rally-task/550ecf7/ : SUCCESS in 44m 17s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/15/670715/3/check/openstack-tox-pylint/57103cf/ : SUCCESS in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/15/670715/3/check/cinder-plugin-ceph-tempest/278e620/ : SUCCESS in 1h 10m 46s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/15/670715/3/check/openstack-tox-functional/120424b/ : SUCCESS in 6m 02s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/15/670715/3/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/c23a108/ : SUCCESS in 1h 08m 13s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/15/670715/3/check/legacy-tempest-dsvm-lvm-multibackend/a9d7e72/ : SUCCESS in 1h 19m 30s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/15/670715/3/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/6000f96/ : SUCCESS in 1h 33m 01s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/15/670715/3/check/tempest-slow-py3/4a45809/ : SUCCESS in 2h 23m 20s","accounts_in_message":[],"_revision_number":3},{"id":"d820745a81e861901c5cab36f89895b264719275","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-23 21:02:32.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/3/check/DellEMC_Unity_FC/434fe5d/DellEMC_Unity_FC/777 : SUCCESS in 1h 09m 40s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/3/check/DellEMC_Unity_iSCSI/527ce0f/DellEMC_Unity_iSCSI/739 : FAILURE in 1h 33m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"16566e7be625f1741c035b256c35069e825a60c9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-23 21:38:12.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/3/check/DellEMC_XtremIO_FC/7621d42/DellEMC_XtremIO_FC/913 : SUCCESS in 1h 28m 56s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/3/check/DellEMC_XtremIO_iSCSI/9a05aa7/DellEMC_XtremIO_iSCSI/851 : SUCCESS in 1h 54m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"2b8cf669dcc08ed1df98e33e7eb83f73c7192522","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-23 22:33:41.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/3/check-not-vote/ext-cinder-zuul/4babe1c : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":3},{"id":"adde64cfaf389b56f03d99dc5ced90286efb48ed","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-24 02:15:09.000000000","message":"Patch Set 3:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/15/670715/3/check-cinder/Cinder-ISER-ISCSI/2dd330e : FAILURE in 3m 24s\n- Cinder-ISER-LIO http://13.74.249.42/15/670715/3/check-cinder/Cinder-ISER-LIO/b315af7 : FAILURE in 3m 37s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"035afba2fc4cfc471dd1e535c973b677167f2c25","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-07-24 08:32:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8e104d60e57bc3c6e0dca33379dbee73ba6be23c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-24 08:41:03.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/4/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/4569652 : FAILURE in 7m 04s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/4/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/d74cecf : FAILURE in 6m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"330c45f69fbf2c0b26693c1bc04a88dc85fd62a2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-24 08:47:37.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a09025ed62da4690c9a5689e67e7159faf694705","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-24 08:52:32.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/4/check/tempest-dsvm-full-xenial/10a3547/ : FAILURE in 14m 51s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/4/check/devstack-plugin-ceph-dsvm-xenial/f5f4a89/ : FAILURE in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e69271a33e0422e1f54cc077e4d9e2f59d3a1bff","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-24 08:55:18.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/4/iscsi/ : FAILURE in 22m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/4/smb/ : FAILURE in 22m 03s","accounts_in_message":[],"_revision_number":4},{"id":"7f2c43367d23a1e5eefdca7aa3971d1c2fabfc18","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-24 08:59:14.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/4/check/dsvm-tempest-infinibox-fc/ab1b06f : FAILURE in 12m 52s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/4/check/dsvm-tempest-infinibox-iscsi/0c8777f : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"e0ebfa7e6f7feb8d1419b3a04c9bd63ff3e2d1cc","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-24 09:04:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/4/check/fujitsu-eternus-dx-iscsi/e637ed0 : FAILURE in 30m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/4/check/fujitsu-eternus-dx-fc/f34e040 : FAILURE in 30m 29s","accounts_in_message":[],"_revision_number":4},{"id":"f39a710e4476c182ded80ff3fb54b912bc2e3484","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-24 09:11:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/4/check/dsvm-tempest-zadara-driver/3b5a033 : FAILURE in 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b5b74051a6bd237b306fb9fb92e80ceb81ef885","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 09:23:26.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/4/check/NEC-ISCSI/bdb5511 : FAILURE in 16m 04s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/4/check/NEC-FC/9a04276 : FAILURE in 15m 47s","accounts_in_message":[],"_revision_number":4},{"id":"86bb3ef425447116f15ad414f3482771babecf80","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-24 09:28:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1836 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1797 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1988 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"35ec543f3c7f90a522b9b8604a029e2279fa7ee8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 09:36:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/4/check/dsvm-tempest-kvm/2a1727c : FAILURE in 18m 08s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/4/check/dsvm-tempest-exe/ba36f20 : FAILURE in 17m 31s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"d1b21186283f46e05d5cad05aa8cc3ab8d93ceac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-24 09:44:55.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/4/check/DellEMC_Unity_FC/dda96aa/DellEMC_Unity_FC/785 : FAILURE in 31m 37s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/4/check/DellEMC_Unity_iSCSI/91a4ff9/DellEMC_Unity_iSCSI/747 : FAILURE in 27m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"d1db24cc42652f45d4d58a51ced51a93d5452f28","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-24 10:04:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/4/check/dsvm-tempest-cisco-zonemanager-job/334ea88/ : FAILURE in 18m 37s","accounts_in_message":[],"_revision_number":4},{"id":"54f7c2efbfde35ac88dd39d06439272d81020e32","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 10:16:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/4/check/DellEMC_XtremIO_FC/61a3cba/DellEMC_XtremIO_FC/922 : FAILURE in 36m 15s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/4/check/DellEMC_XtremIO_iSCSI/2c280c6/DellEMC_XtremIO_iSCSI/859 : FAILURE in 35m 15s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"fc8db2b6619e82185e8b17c2b76bc7cd4665fe70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 11:26:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/670715/4/check/openstack-tox-lower-constraints/21e0497/ : SUCCESS in 18m 31s\n- openstack-tox-pep8 http://logs.openstack.org/15/670715/4/check/openstack-tox-pep8/e87cde3/ : SUCCESS in 8m 03s\n- openstack-tox-py27 http://logs.openstack.org/15/670715/4/check/openstack-tox-py27/84a3609/ : FAILURE in 10m 21s\n- openstack-tox-py36 http://logs.openstack.org/15/670715/4/check/openstack-tox-py36/89347c3/ : SUCCESS in 9m 56s\n- openstack-tox-py37 http://logs.openstack.org/15/670715/4/check/openstack-tox-py37/d0e1d68/ : SUCCESS in 9m 10s\n- openstack-tox-docs http://logs.openstack.org/15/670715/4/check/openstack-tox-docs/3d03f7f/ : FAILURE in 8m 11s\n- tempest-full http://logs.openstack.org/15/670715/4/check/tempest-full/3956a52/ : SUCCESS in 1h 32m 16s\n- neutron-grenade http://logs.openstack.org/15/670715/4/check/neutron-grenade/488dfb9/ : SUCCESS in 1h 02m 06s\n- grenade-py3 http://logs.openstack.org/15/670715/4/check/grenade-py3/0b47635/ : SUCCESS in 1h 03m 51s\n- tempest-full-py3 http://logs.openstack.org/15/670715/4/check/tempest-full-py3/b51df15/ : SUCCESS in 1h 29m 24s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/15/670715/4/check/cinder-tempest-dsvm-lvm-lio-barbican/6de3efb/ : SUCCESS in 1h 23m 29s\n- cinder-tox-bandit-baseline http://logs.openstack.org/15/670715/4/check/cinder-tox-bandit-baseline/9967e82/ : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/15/670715/4/check/openstack-tox-functional-py36/c77fa2d/ : SUCCESS in 5m 00s\n- cinder-rally-task http://logs.openstack.org/15/670715/4/check/cinder-rally-task/bd4940f/ : SUCCESS in 42m 21s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/15/670715/4/check/openstack-tox-pylint/b95474d/ : SUCCESS in 4m 28s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/15/670715/4/check/cinder-plugin-ceph-tempest/472b9c6/ : SUCCESS in 1h 15m 05s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/15/670715/4/check/openstack-tox-functional/76d7188/ : SUCCESS in 5m 58s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/15/670715/4/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/a5bb816/ : SUCCESS in 1h 09m 43s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/15/670715/4/check/legacy-tempest-dsvm-lvm-multibackend/e0f8cba/ : SUCCESS in 1h 26m 20s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/15/670715/4/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/d02f7ec/ : SUCCESS in 1h 29m 05s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/15/670715/4/check/tempest-slow-py3/a199b03/ : SUCCESS in 2h 26m 30s","accounts_in_message":[],"_revision_number":4},{"id":"718a1dfdcf361b58f2cd71ab93642cee9213f561","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-07-24 13:34:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"14336fadf56136799f9077f0ce6c6a666700bc04","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-24 13:54:41.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1843 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1807 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1995 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"64e0b0ebc25bf516e5a62f8e4a3681a56c0d3a60","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 14:18:08.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/5/check/dsvm-tempest-kvm/47e022d : FAILURE in 37m 04s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/5/check/dsvm-tempest-exe/750d7b2 : FAILURE in 37m 50s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"8de6e71b7ca67f9f22cec653475e95e6e5a16b20","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 14:18:50.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/5/check/NEC-ISCSI/78e5a9c : FAILURE in 40m 57s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/5/check/NEC-FC/04defdd : FAILURE in 41m 57s","accounts_in_message":[],"_revision_number":5},{"id":"2d64f2de4263c255068c058ad0b02d5cee0c58d5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-24 14:29:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/5/iscsi/ : SUCCESS in 54m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/5/smb/ : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":5},{"id":"a7d555261ba83744d93774d0dea76268c653f49b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-24 14:37:49.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/5/check/dsvm-tempest-infinibox-fc/bebf882 : SUCCESS in 55m 44s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/5/check/dsvm-tempest-infinibox-iscsi/25a7b70 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"ba028e345f3f4529bf8689e2fb41602ebcccb786","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-24 15:12:43.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/5/check/tempest-dsvm-full-xenial/0d010c1/ : SUCCESS in 1h 24m 30s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/5/check/devstack-plugin-ceph-dsvm-xenial/1671b52/ : SUCCESS in 1h 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a120b72b337cfa46bccf82eb11be853ba5fd5023","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-24 15:14:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/5/check/dsvm-tempest-zadara-driver/4589acb : SUCCESS in 1h 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"05c598a762cf2436ccdd20f9b086216e8ad15079","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-24 15:20:46.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"ebe112b32c2be7a077c2da7359fafe111f295569","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-24 15:24:15.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/5/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1c14c3f : SUCCESS in 48m 10s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/5/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/99ab68b : FAILURE in 1h 49m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"401fe67d0ea9ba2535844564c1f43319aa10b400","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-24 15:31:41.000000000","message":"Patch Set 5:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/5/check/fujitsu-eternus-dx-iscsi/f259a74 : SUCCESS in 1h 55m 43s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/5/check/fujitsu-eternus-dx-fc/475e759 : SUCCESS in 1h 56m 23s","accounts_in_message":[],"_revision_number":5},{"id":"2c883e1c1b508c7dfcc987f524254c121ee5ab28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 16:27:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/670715/5/check/openstack-tox-lower-constraints/fa72e4b/ : SUCCESS in 13m 17s\n- openstack-tox-pep8 http://logs.openstack.org/15/670715/5/check/openstack-tox-pep8/163bb26/ : SUCCESS in 9m 14s\n- openstack-tox-py27 http://logs.openstack.org/15/670715/5/check/openstack-tox-py27/bb95887/ : SUCCESS in 9m 48s\n- openstack-tox-py36 http://logs.openstack.org/15/670715/5/check/openstack-tox-py36/4a77563/ : SUCCESS in 9m 17s\n- openstack-tox-py37 http://logs.openstack.org/15/670715/5/check/openstack-tox-py37/ad90c8b/ : SUCCESS in 9m 24s\n- openstack-tox-docs http://logs.openstack.org/15/670715/5/check/openstack-tox-docs/1607e2b/ : FAILURE in 8m 33s\n- tempest-full http://logs.openstack.org/15/670715/5/check/tempest-full/35c1461/ : FAILURE in 1h 31m 26s\n- neutron-grenade http://logs.openstack.org/15/670715/5/check/neutron-grenade/bec5e8c/ : FAILURE in 1h 27m 59s\n- grenade-py3 http://logs.openstack.org/15/670715/5/check/grenade-py3/3bd390c/ : SUCCESS in 57m 59s\n- tempest-full-py3 http://logs.openstack.org/15/670715/5/check/tempest-full-py3/9699515/ : SUCCESS in 1h 25m 44s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/15/670715/5/check/cinder-tempest-dsvm-lvm-lio-barbican/6013baf/ : FAILURE in 1h 35m 53s\n- cinder-tox-bandit-baseline http://logs.openstack.org/15/670715/5/check/cinder-tox-bandit-baseline/32e93cf/ : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/15/670715/5/check/openstack-tox-functional-py36/1cd5543/ : SUCCESS in 6m 21s\n- cinder-rally-task http://logs.openstack.org/15/670715/5/check/cinder-rally-task/4c3c34c/ : SUCCESS in 45m 44s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/15/670715/5/check/openstack-tox-pylint/4b7fb19/ : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/15/670715/5/check/cinder-plugin-ceph-tempest/a3cd1fb/ : SUCCESS in 1h 15m 57s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/15/670715/5/check/openstack-tox-functional/722e6c3/ : SUCCESS in 8m 56s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/15/670715/5/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/f0626fa/ : SUCCESS in 1h 24m 43s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/15/670715/5/check/legacy-tempest-dsvm-lvm-multibackend/3aef03f/ : FAILURE in 1h 46m 38s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/15/670715/5/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/af43a28/ : SUCCESS in 1h 23m 40s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/15/670715/5/check/tempest-slow-py3/10048d7/ : SUCCESS in 2h 33m 26s","accounts_in_message":[],"_revision_number":5},{"id":"7a7b34c3150a84ce10536ecceda6f3605bd33812","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-24 16:30:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/5/check/dsvm-tempest-cisco-zonemanager-job/6274af6/ : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":5},{"id":"8b75c8ff573fea57ec247cccdc5e9d3c44aa1624","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-24 20:40:40.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/5/check/DellEMC_Unity_FC/09a48b6/DellEMC_Unity_FC/796 : FAILURE in 55m 09s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/5/check/DellEMC_Unity_iSCSI/4cd6a47/DellEMC_Unity_iSCSI/758 : FAILURE in 1h 08m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b3f7697b7b0a989281a7fc453cec9e5e70e21344","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 01:07:15.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/5/check/DellEMC_XtremIO_FC/800879a/DellEMC_XtremIO_FC/932 : SUCCESS in 1h 40m 34s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/5/check/DellEMC_XtremIO_iSCSI/931dcfa/DellEMC_XtremIO_iSCSI/869 : SUCCESS in 1h 51m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"948b441bed1d7a627b3fe761766902d998f8c6e4","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-07-25 05:54:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f170049b656dee8a0b51e40e191647e34b0de924","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-25 06:02:52.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/6/check/fujitsu-eternus-dx-iscsi/37b4809 : FAILURE in 7m 23s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/6/check/fujitsu-eternus-dx-fc/9c94163 : FAILURE in 7m 12s","accounts_in_message":[],"_revision_number":6},{"id":"cf8f4e51e8afc5a29a033db1fec0fc3eb0ecd807","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-25 06:07:00.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"791af629d5420d8651cf7ff82ef8415c21b709da","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-25 06:09:55.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/6/iscsi/ : FAILURE in 14m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/6/smb/ : FAILURE in 14m 29s","accounts_in_message":[],"_revision_number":6},{"id":"55ebfed728bf9428ec3f7e626ec452c6e9986889","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-25 06:13:06.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/6/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/88c5ebc : FAILURE in 18m 03s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/6/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/4e172b8 : FAILURE in 17m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"304237756a6ca04a9d3990ed159cba9d4ce22c57","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-25 06:15:59.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/6/check/tempest-dsvm-full-xenial/9134251/ : FAILURE in 20m 51s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/6/check/devstack-plugin-ceph-dsvm-xenial/90d4a1d/ : FAILURE in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"681718a78d425a9b41a690c7059d951ea1858748","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-25 06:16:07.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1849 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2001 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1813 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"4c6d294d3196c97b6462786c7eba46c16c6f8ffd","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 06:26:58.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/6/check/NEC-ISCSI/5c69e69 : FAILURE in 31m 43s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/6/check/NEC-FC/2fecf62 : FAILURE in 30m 17s","accounts_in_message":[],"_revision_number":6},{"id":"87359511ea0d6888e7bd9460c0370f14cb9ea78d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 06:37:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/6/check/dsvm-tempest-kvm/30e32e9 : FAILURE in 35m 47s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/6/check/dsvm-tempest-exe/a17fa58 : FAILURE in 36m 07s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"5045b4299641d93f7c463cd8fcfc05502c199145","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-25 06:41:14.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/6/check/dsvm-tempest-cisco-zonemanager-job/0578cce/ : FAILURE in 41m 12s","accounts_in_message":[],"_revision_number":6},{"id":"b6f77aa00e76187587bf957dae2703f01919aaec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-25 06:46:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/6/check/DellEMC_Unity_FC/742c118/DellEMC_Unity_FC/804 : FAILURE in 51m 16s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/6/check/DellEMC_Unity_iSCSI/0b918d0/DellEMC_Unity_iSCSI/765 : FAILURE in 46m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"699aa507160f462ea5c169d44d34f65a02523f56","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-25 06:48:57.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/6/check/dsvm-tempest-infinibox-fc/591e6f0 : SUCCESS in 53m 35s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/6/check/dsvm-tempest-infinibox-iscsi/5677411 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"c523049cd02ae19768348ec3652baeafb9dcb3d9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-25 06:55:20.000000000","message":"Patch Set 6:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/15/670715/6/check-cinder/Cinder-ISER-ISCSI/a889e61 : FAILURE in 26m 37s\n- Cinder-ISER-LIO http://13.74.249.42/15/670715/6/check-cinder/Cinder-ISER-LIO/f2acc7e : FAILURE in 28m 45s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"9df9070dd244351a9d73c1782f61e848a680770a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 07:15:47.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/15/670715/6/check/DellEMC_PowerMAX_FC/12be2d1/DellEMC_PowerMAX_FC/488 : FAILURE in 2m 29s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/15/670715/6/check/DellEMC_PowerMAX_iSCSI/b0b0b59/DellEMC_PowerMAX_iSCSI/472 : FAILURE in 2m 25s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"72ff50780ad358b57fa95ca58f45a53821a62c80","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 07:16:20.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/6/check/DellEMC_XtremIO_FC/a19286b/DellEMC_XtremIO_FC/938 : FAILURE in 3m 52s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/6/check/DellEMC_XtremIO_iSCSI/2df086e/DellEMC_XtremIO_iSCSI/875 : FAILURE in 2m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"3c86a7bb4a165d9c4a458d07c0aae722133c661c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 07:30:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/670715/6/check/openstack-tox-lower-constraints/d0c97c2/ : SUCCESS in 11m 26s\n- openstack-tox-pep8 http://logs.openstack.org/15/670715/6/check/openstack-tox-pep8/228a39a/ : SUCCESS in 8m 22s\n- openstack-tox-py27 http://logs.openstack.org/15/670715/6/check/openstack-tox-py27/229f748/ : SUCCESS in 10m 37s\n- openstack-tox-py36 http://logs.openstack.org/15/670715/6/check/openstack-tox-py36/666ae18/ : SUCCESS in 9m 42s\n- openstack-tox-py37 http://logs.openstack.org/15/670715/6/check/openstack-tox-py37/c5793be/ : SUCCESS in 9m 42s\n- openstack-tox-docs http://logs.openstack.org/15/670715/6/check/openstack-tox-docs/7066325/html/ : SUCCESS in 9m 00s\n- tempest-full http://logs.openstack.org/15/670715/6/check/tempest-full/1143ded/ : FAILURE in 26m 10s\n- neutron-grenade http://logs.openstack.org/15/670715/6/check/neutron-grenade/361b9c0/ : FAILURE in 36m 25s\n- grenade-py3 http://logs.openstack.org/15/670715/6/check/grenade-py3/d467a4b/ : FAILURE in 37m 23s\n- tempest-full-py3 http://logs.openstack.org/15/670715/6/check/tempest-full-py3/cdec56a/ : SUCCESS in 1h 33m 26s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/15/670715/6/check/cinder-tempest-dsvm-lvm-lio-barbican/f00824e/ : FAILURE in 29m 33s\n- cinder-tox-bandit-baseline http://logs.openstack.org/15/670715/6/check/cinder-tox-bandit-baseline/8ccac63/ : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/15/670715/6/check/openstack-tox-functional-py36/e6ef959/ : SUCCESS in 5m 35s\n- cinder-rally-task http://logs.openstack.org/15/670715/6/check/cinder-rally-task/b22d97a/ : SUCCESS in 46m 17s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/15/670715/6/check/openstack-tox-pylint/5f62aa3/ : SUCCESS in 7m 19s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/15/670715/6/check/cinder-plugin-ceph-tempest/1b73d05/ : FAILURE in 28m 15s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/15/670715/6/check/openstack-tox-functional/14de392/ : SUCCESS in 5m 40s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/15/670715/6/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/d64629b/ : FAILURE in 36m 39s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/15/670715/6/check/legacy-tempest-dsvm-lvm-multibackend/cde5c88/ : FAILURE in 31m 21s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/15/670715/6/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/16a5d6f/ : FAILURE in 26m 04s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/15/670715/6/check/tempest-slow-py3/aa4c9ff/ : FAILURE in 28m 53s","accounts_in_message":[],"_revision_number":6},{"id":"d43dab8250fa34b6631f876c7fd7a210f899ba73","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-25 07:40:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/6/check/dsvm-tempest-zadara-driver/7544de7 : FAILURE in 1h 45m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eacfa477a92199fa174390323360fcf5478d4be0","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-07-25 12:10:58.000000000","message":"Patch Set 6:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":6},{"id":"65a208056ceb4490e832a50ae501b065e227b78b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-25 13:26:03.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"30b92921497a309a76459de60c2632cb98fd3d57","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-07-25 13:39:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f92a3d36178cf541ca950c9135d09a3b9cb73165","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-25 13:40:46.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/7/check/fujitsu-eternus-dx-iscsi/9902833 : FAILURE in 36s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/7/check/fujitsu-eternus-dx-fc/2743da4 : FAILURE in 36s","accounts_in_message":[],"_revision_number":7},{"id":"ee933091334e1e3dc41797b1bed4b3a746da0ba2","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-25 14:00:42.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1852 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1816 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2004 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"f33274e903d6351a5c16c6c3e8886bf476515b89","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 14:07:17.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/7/check/NEC-ISCSI/c6291cd : FAILURE in 24m 08s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/7/check/NEC-FC/276a5f7 : FAILURE in 23m 10s","accounts_in_message":[],"_revision_number":7},{"id":"16f48e2e0b0d590d044eb077f452be6154333c4f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 14:27:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/7/check/dsvm-tempest-kvm/fe1b34a : FAILURE in 32m 23s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/7/check/dsvm-tempest-exe/b7fa24e : FAILURE in 41m 16s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"0e288dac4013219422e77c48203abb2a241eaba6","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-25 14:36:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/7/iscsi/ : SUCCESS in 56m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/7/smb/ : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":7},{"id":"b086e06ac63bd94488345b0127a58ae673e9faf3","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-25 14:45:25.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/7/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1557700 : SUCCESS in 56m 33s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/7/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/1f02b47 : SUCCESS in 1h 05m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"a10c82ea658dcf1898ac4499edba041a3d28f5c5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-25 14:52:11.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/7/check/dsvm-tempest-infinibox-fc/c0f838e : SUCCESS in 56m 56s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/7/check/dsvm-tempest-infinibox-iscsi/005b6b7 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"f07e8212dc691dc37736a50c698ca9d6caac7393","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-25 14:52:15.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/7/check/DellEMC_Unity_FC/2bfc952/DellEMC_Unity_FC/808 : FAILURE in 50m 25s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/7/check/DellEMC_Unity_iSCSI/01a28a2/DellEMC_Unity_iSCSI/769 : FAILURE in 49m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"6da0cd8a7506d61f4433a41fe855124b91f0c22f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-25 15:20:52.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"9acecd11881a73f7d1379aa676ab316fe9d256ec","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-25 15:32:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/7/check/dsvm-tempest-zadara-driver/bc72c2d : SUCCESS in 1h 40m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cd32c8c05db0f226da020755c6337db6640cbbff","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-25 16:01:01.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/7/check/dsvm-tempest-cisco-zonemanager-job/885f7fc/ : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":7},{"id":"eb4e3f6b7a5576f6ae659e4eb99b21fb12fd587a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-25 16:10:25.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/7/check/tempest-dsvm-full-xenial/999f14a/ : SUCCESS in 1h 27m 49s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/7/check/devstack-plugin-ceph-dsvm-xenial/5fb6354/ : FAILURE in 1h 48m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b29e4cdf2af3f93d4ad6099a99cd9980f4b878bd","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 17:23:22.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/7/check/DellEMC_XtremIO_FC/0bcdb12/DellEMC_XtremIO_FC/942 : SUCCESS in 1h 27m 40s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/7/check/DellEMC_XtremIO_iSCSI/15d50bb/DellEMC_XtremIO_iSCSI/879 : SUCCESS in 1h 37m 06s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"4e34e4384314c6415b628f630dc96af940fef95c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 17:40:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/15/670715/7/check/openstack-tox-lower-constraints/ce7f42e/ : SUCCESS in 13m 00s\n- openstack-tox-pep8 http://logs.openstack.org/15/670715/7/check/openstack-tox-pep8/ff5bba2/ : SUCCESS in 9m 05s\n- openstack-tox-py27 http://logs.openstack.org/15/670715/7/check/openstack-tox-py27/1a17aef/ : SUCCESS in 7m 50s\n- openstack-tox-py36 http://logs.openstack.org/15/670715/7/check/openstack-tox-py36/8e4079d/ : SUCCESS in 11m 24s\n- openstack-tox-py37 http://logs.openstack.org/15/670715/7/check/openstack-tox-py37/6faa7fd/ : SUCCESS in 9m 26s\n- openstack-tox-docs http://logs.openstack.org/15/670715/7/check/openstack-tox-docs/50cbd99/html/ : SUCCESS in 10m 05s\n- tempest-full http://logs.openstack.org/15/670715/7/check/tempest-full/beac0c0/ : SUCCESS in 1h 35m 43s\n- neutron-grenade http://logs.openstack.org/15/670715/7/check/neutron-grenade/2837ae9/ : SUCCESS in 1h 02m 37s\n- grenade-py3 http://logs.openstack.org/15/670715/7/check/grenade-py3/06c190a/ : SUCCESS in 1h 00m 09s\n- tempest-full-py3 http://logs.openstack.org/15/670715/7/check/tempest-full-py3/9b9edca/ : SUCCESS in 1h 23m 44s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/15/670715/7/check/cinder-tempest-dsvm-lvm-lio-barbican/0a0eab4/ : SUCCESS in 1h 44m 14s\n- cinder-tox-bandit-baseline http://logs.openstack.org/15/670715/7/check/cinder-tox-bandit-baseline/ca51d42/ : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/15/670715/7/check/openstack-tox-functional-py36/6020f1e/ : SUCCESS in 4m 48s\n- cinder-rally-task http://logs.openstack.org/15/670715/7/check/cinder-rally-task/2f0a763/ : SUCCESS in 44m 42s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/15/670715/7/check/openstack-tox-pylint/7487f83/ : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/15/670715/7/check/cinder-plugin-ceph-tempest/e174f7b/ : SUCCESS in 1h 17m 34s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/15/670715/7/check/openstack-tox-functional/84ef82b/ : SUCCESS in 6m 46s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/15/670715/7/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/25771b4/ : SUCCESS in 1h 05m 37s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/15/670715/7/check/legacy-tempest-dsvm-lvm-multibackend/b64e389/ : SUCCESS in 1h 26m 45s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/15/670715/7/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/a77d425/ : SUCCESS in 1h 37m 39s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/15/670715/7/check/tempest-slow-py3/5b9b978/ : SUCCESS in 2h 37m 24s","accounts_in_message":[],"_revision_number":7},{"id":"2fb654719a1de93d3f34f94108cc583a46f01542","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-25 17:51:14.000000000","message":"Patch Set 7:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-670715-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"afce11a6ab480d2be515957704bbd57f86d8cf72","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-25 18:28:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/7/check/DellEMC_Unity_FC/5eee33c/DellEMC_Unity_FC/810 : FAILURE in 45m 47s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/7/check/DellEMC_Unity_iSCSI/2054d5a/DellEMC_Unity_iSCSI/771 : FAILURE in 47m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"052bfb1b679b2aeced2b3f2cbe1175df69c31849","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-07-25 18:34:53.000000000","message":"Patch Set 7:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/15/2419 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"995259fceffc5d9d4e2304fbd650b5d96d3b9b4b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-25 18:39:39.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/7/check-not-vote/ext-cinder-zuul/d3d910a : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":7},{"id":"ed835160950fa70e2257058f90d487b50de49ad3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-25 18:55:24.000000000","message":"Patch Set 7:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-15-670715-7/ns5_iscsi : FAILURE in 22m 23s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-15-670715-7/ns5_nfs : SUCCESS in 70m 55s","accounts_in_message":[],"_revision_number":7},{"id":"25fe0d3b3a5f93e8d6b4e2369c6e9fc3546342b9","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 19:01:36.000000000","message":"Patch Set 7:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-15-670715-7-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"4655b4e48111a5bf31df61963c1a03a3d85b7cbe","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-07-25 19:15:33.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/670715/7/check/DellEMC_SC_FC/47449f2/DellEMC_SC_FC/332 : SUCCESS in 1h 14m 18s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/670715/7/check/DellEMC_SC_iSCSI/24c635b/DellEMC_SC_iSCSI/315 : SUCCESS in 1h 34m 25s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"50d6118f929b22c9d0ebedfd1f27dcb165c3505d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 19:19:15.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/15/670715/7/check/DellEMC_VNX_iSCSI/b6506aa/DellEMC_VNX_iSCSI/468 : SUCCESS in 1h 38m 06s\n- DellEMC_VNX_FC http://publiclogs.emc.com/15/670715/7/check/DellEMC_VNX_FC/3a60f98/DellEMC_VNX_FC/478 : FAILURE in 1h 07m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"5c40a8fc3f7288b92f5f28ae3ded3761c0a56a78","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 19:21:53.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/15/670715/7/check/DellEMC_PowerMAX_FC/d2e7395/DellEMC_PowerMAX_FC/492 : FAILURE in 2h 29m 09s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/15/670715/7/check/DellEMC_PowerMAX_iSCSI/ea227ce/DellEMC_PowerMAX_iSCSI/476 : FAILURE in 2h 15m 30s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"f54773dc00a383d6ca4011bb012d775c346d8c27","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-07-25 19:32:02.000000000","message":"Patch Set 7:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/670715-7/HPMSA-FC : FAILURE in 37:12.66\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/670715-7/HPMSA-iSCSI : FAILURE in 36:50.21","accounts_in_message":[],"_revision_number":7},{"id":"82ccb2c58b3b117da08a24362d0e2660789fda6e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 19:36:12.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/15/670715/7/check/DellEMC_VxFlexOS/44bffa5/DellEMC_VxFlexOS/499 : SUCCESS in 1h 55m 04s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"f94701f778f3de2d7bfd3315c51c0791412538fa","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-07-25 19:46:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/15/670715/7/check/kaminario-dsvm-tempest-full-iscsi/91f9bd1 : SUCCESS in 1h 56m 00s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/15/670715/7/check/kaminario-dsvm-tempest-full-FC/3cb4a76 : SUCCESS in 2h 02m 40s","accounts_in_message":[],"_revision_number":7},{"id":"74828418e6357c756999eab6a782f450bd17b653","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-07-25 20:05:34.000000000","message":"Patch Set 7:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/670715-7/Lenovo-FC : FAILURE in 35:10.48\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/670715-7/Lenovo-iSCSI : FAILURE in 1:10:00","accounts_in_message":[],"_revision_number":7},{"id":"ca76582451c27fa23bead02c4ce77978f2375bbe","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-25 20:12:05.000000000","message":"Patch Set 7:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c7108e4 : SUCCESS in 2h 10m 06s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/7/check/3par-fc-driver-master-client-pip-ssa02-dsvm/8dfce4f : FAILURE in 2h 28m 41s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/15/670715/7/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/a2fe5bc : FAILURE in 1h 42m 28s","accounts_in_message":[],"_revision_number":7},{"id":"b011b32595cbca51c74c7a4dc43aad4063bcc878","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-25 21:06:37.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/7/check/DellEMC_XtremIO_FC/bd80516/DellEMC_XtremIO_FC/944 : SUCCESS in 1h 26m 33s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/7/check/DellEMC_XtremIO_iSCSI/f798a6f/DellEMC_XtremIO_iSCSI/881 : SUCCESS in 1h 53m 49s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"231eec89ecba59086b49c6fa9bb7b39188cc582b","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-08-05 15:12:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"efa71bf948f368765fcdf3929b0e6353b2843d83","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-05 15:36:45.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1910 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1927 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2078 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"33647c0bd2d4ccca14ea303234edecb25cda2f6b","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-05 15:57:03.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/8/check/dsvm-tempest-kvm/ebce56c : FAILURE in 35m 42s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/8/check/dsvm-tempest-exe/8ef532c : FAILURE in 34m 44s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"e8981b37d99e8f3072e91d39336f2e3943500f30","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-05 16:08:19.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/8/iscsi/ : SUCCESS in 54m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/8/smb/ : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":8},{"id":"57cc19700760621c5dc4f24c4a6378aa356d2124","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-05 16:22:09.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/8/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/46eab5e : FAILURE in 53m 56s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/8/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/b732a49 : FAILURE in 1h 09m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"c9c3aa155b416344e819bd02871c6b921d9c0c5e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-05 16:22:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/8/check/dsvm-tempest-cisco-zonemanager-job/b19a5b9/ : SUCCESS in 1h 04m 34s","accounts_in_message":[],"_revision_number":8},{"id":"fc53bf286c10bf0338ce9298f99ac2dad8a8a8cf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-05 16:25:47.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/8/check/dsvm-tempest-infinibox-fc/7b8b21b : FAILURE in 54m 30s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/8/check/dsvm-tempest-infinibox-iscsi/2f8e2d9 : FAILURE in 1h 12m 26s","accounts_in_message":[],"_revision_number":8},{"id":"721b5325dac1d03144dc4aaa96c297749b0e9cba","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-05 16:49:19.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/8/check-not-vote/ext-cinder-zuul/4243be6 : FAILURE in 16m 22s","accounts_in_message":[],"_revision_number":8},{"id":"030dfcfb9134a43b9756e1c185e04518f8611620","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-05 16:51:21.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/8/check/fujitsu-eternus-dx-iscsi/f68e9d3 : FAILURE in 1h 37m 54s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/8/check/fujitsu-eternus-dx-fc/5f4c10a : FAILURE in 2m 27s","accounts_in_message":[],"_revision_number":8},{"id":"13813b80cf309553d43576d60c1766d1805aa1b7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-05 16:51:58.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/8/check/dsvm-tempest-zadara-driver/c0ac8e0 : FAILURE in 59m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"273d3678ae82d2bb258a9ccfb70ee0ddc82ad577","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-05 16:52:05.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"b4053ee84beb1c2b7fb6589a15fe28fb7e825830","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-05 17:09:01.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/8/check/tempest-dsvm-full-xenial/3191865/ : SUCCESS in 1h 50m 07s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/8/check/devstack-plugin-ceph-dsvm-xenial/bb1c83e/ : SUCCESS in 1h 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"10c83296b763d40ef47238d42112401ff021da90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:48:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/15/670715/8/check/openstack-tox-lower-constraints/2a8aee2/ : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://logs.opendev.org/15/670715/8/check/openstack-tox-pep8/dc434e8/ : SUCCESS in 12m 11s\n- openstack-tox-py27 https://logs.opendev.org/15/670715/8/check/openstack-tox-py27/402cd96/ : SUCCESS in 10m 22s\n- openstack-tox-py36 https://logs.opendev.org/15/670715/8/check/openstack-tox-py36/0cb7069/ : SUCCESS in 10m 53s\n- openstack-tox-py37 https://logs.opendev.org/15/670715/8/check/openstack-tox-py37/68fa497/ : SUCCESS in 10m 03s\n- openstack-tox-docs https://logs.opendev.org/15/670715/8/check/openstack-tox-docs/5ebc49b/html/ : SUCCESS in 9m 14s\n- tempest-full https://logs.opendev.org/15/670715/8/check/tempest-full/b411562/ : FAILURE in 1h 32m 13s\n- neutron-grenade https://logs.opendev.org/15/670715/8/check/neutron-grenade/46c9aab/ : SUCCESS in 1h 13m 22s\n- grenade-py3 https://logs.opendev.org/15/670715/8/check/grenade-py3/b96d82a/ : SUCCESS in 57m 52s\n- tempest-integrated-storage https://logs.opendev.org/15/670715/8/check/tempest-integrated-storage/c5bb5f9/ : SUCCESS in 1h 24m 07s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/15/670715/8/check/cinder-tempest-dsvm-lvm-lio-barbican/b375832/ : SUCCESS in 1h 43m 03s\n- cinder-tox-bandit-baseline https://logs.opendev.org/15/670715/8/check/cinder-tox-bandit-baseline/893aa38/ : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/15/670715/8/check/openstack-tox-functional-py36/7a3fd9c/ : SUCCESS in 5m 41s\n- cinder-rally-task https://logs.opendev.org/15/670715/8/check/cinder-rally-task/98611ab/ : SUCCESS in 39m 38s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/15/670715/8/check/openstack-tox-pylint/0123db0/ : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/15/670715/8/check/cinder-plugin-ceph-tempest/efc06d3/ : SUCCESS in 1h 17m 56s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/15/670715/8/check/openstack-tox-functional/1e8caa4/ : SUCCESS in 7m 22s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/15/670715/8/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/01ef666/ : SUCCESS in 1h 28m 34s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/15/670715/8/check/legacy-tempest-dsvm-lvm-multibackend/d488e79/ : FAILURE in 1h 23m 23s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/15/670715/8/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/d9c0d6f/ : SUCCESS in 1h 21m 10s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/15/670715/8/check/tempest-slow-py3/76ea1c4/ : SUCCESS in 2h 31m 43s","accounts_in_message":[],"_revision_number":8},{"id":"d0f95eebed2f5543381363fc3fab70a2ff1e4c39","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-05 18:07:17.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/8/check/DellEMC_Unity_FC/d10bfaf/DellEMC_Unity_FC/918 : FAILURE in 48m 43s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/8/check/DellEMC_Unity_iSCSI/8150494/DellEMC_Unity_iSCSI/877 : FAILURE in 48m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"c209d93401f27f493c309bed830d4991d6121f34","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-05 18:22:32.000000000","message":"Patch Set 8:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/15/670715/8/check-cinder/Cinder-ISER-ISCSI/a65ce07 : FAILURE in 1h 01m 38s\n- Cinder-ISER-LIO http://13.74.249.42/15/670715/8/check-cinder/Cinder-ISER-LIO/9a041f5 : SUCCESS in 1h 09m 28s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"6d24c07002e74ef6461c4545e1a8eec82c8a05d4","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-05 18:58:04.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/8/check/NEC-ISCSI/9ee0640 : FAILURE in 1h 52m 54s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/8/check/NEC-FC/3a3b0b6 : FAILURE in 1h 36m 01s","accounts_in_message":[],"_revision_number":8},{"id":"9ab5defcd39ca27a6070cf49a1852c2d482684a1","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-06 01:55:45.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"69f09f1329b3d7bdc4f4f4c9c732fd36f5d07585","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-06 02:00:56.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/8/check/DellEMC_XtremIO_FC/2ebe70a/DellEMC_XtremIO_FC/1042 : FAILURE in 1h 28m 08s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/8/check/DellEMC_XtremIO_iSCSI/8b9c90b/DellEMC_XtremIO_iSCSI/971 : FAILURE in 3h 52m 02s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"246bec47c5cbf53cf1f6b08e73027d76187a8915","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-08-06 13:45:15.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5940933319562032b75c2871425ee335e350d8f8","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-06 14:11:40.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1923 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1938 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2089 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"7186383257263511e4f686a4cec25e00d820249c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-06 14:46:55.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/9/check/dsvm-tempest-kvm/65310e5 : FAILURE in 36m 14s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/9/check/dsvm-tempest-exe/6b5cf59 : FAILURE in 35m 20s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"17bfbf84baae974e1657e55db2b783c2efa09055","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-06 14:47:14.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/9/iscsi/ : SUCCESS in 59m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/9/smb/ : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":9},{"id":"46c25b3105a72f069132cbab3854f2de3ba89740","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-06 14:54:32.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/9/check/dsvm-tempest-infinibox-fc/346fbb4 : FAILURE in 1h 00m 27s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/9/check/dsvm-tempest-infinibox-iscsi/afc49b3 : FAILURE in 1h 08m 46s","accounts_in_message":[],"_revision_number":9},{"id":"b1a77634f00e8f27ccd24d458913e709a77cb1e8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-06 14:57:04.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/9/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/5388933 : FAILURE in 1h 00m 50s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/9/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/6d97eed : FAILURE in 1h 11m 28s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"40d257dd9d693d0be6216de3dbb21d810e06f8d0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-06 15:18:17.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/9/check/DellEMC_Unity_FC/a4e93da/DellEMC_Unity_FC/933 : FAILURE in 1h 08m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/9/check/DellEMC_Unity_iSCSI/0457e73/DellEMC_Unity_iSCSI/892 : FAILURE in 1h 25m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"150c3b0025afe6a4add6e2b227bd1314c947bf71","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-06 15:18:19.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"800ff9bce44844422888d38fdba657a6fd98c7f9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-06 15:20:42.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/9/check/DellEMC_XtremIO_FC/6ba971a/DellEMC_XtremIO_FC/1055 : FAILURE in 1h 27m 07s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/9/check/DellEMC_XtremIO_iSCSI/c882258/DellEMC_XtremIO_iSCSI/978 : FAILURE in 1h 26m 33s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"6f917ecd55fed29111bb38ac1462b08b269ad29c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-06 15:23:29.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/9/check/fujitsu-eternus-dx-iscsi/2a6438a : FAILURE in 1h 37m 33s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/9/check/fujitsu-eternus-dx-fc/1516f9b : FAILURE in 2m 26s","accounts_in_message":[],"_revision_number":9},{"id":"2a7d496f99d0302245a28ff56c1f6c0b51beb356","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-06 16:01:47.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/9/check/dsvm-tempest-zadara-driver/6d1f114 : FAILURE in 1h 45m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b61226519b3ab3c6601a79cfa59a054ee52a71c3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-06 16:42:42.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/9/check/NEC-ISCSI/15648a8 : FAILURE in 38m 25s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/9/check/NEC-FC/1d1bceb : FAILURE in 1h 57m 21s","accounts_in_message":[],"_revision_number":9},{"id":"6eca521a9ab772c18d5eaa8a995162afa8c19294","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-06 18:10:58.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/9/check-not-vote/ext-cinder-zuul/6f6deb9 : FAILURE in 1m 51s","accounts_in_message":[],"_revision_number":9},{"id":"bb099856fbf8fe954e7cf4be13657d478a28bf54","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-06 19:17:03.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/9/check/tempest-dsvm-full-xenial/2592ce0/ : FAILURE in 1h 27m 21s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/9/check/devstack-plugin-ceph-dsvm-xenial/b4e22e2/ : FAILURE in 1h 58m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"08e6746cbb4d1a1c9369ca7f2fb4e163906111b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 20:24:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/15/670715/9/check/openstack-tox-lower-constraints/0434979/ : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://logs.opendev.org/15/670715/9/check/openstack-tox-pep8/ce972b5/ : SUCCESS in 9m 25s\n- openstack-tox-py27 https://logs.opendev.org/15/670715/9/check/openstack-tox-py27/241c300/ : SUCCESS in 9m 57s\n- openstack-tox-py36 https://logs.opendev.org/15/670715/9/check/openstack-tox-py36/203baf5/ : FAILURE in 15m 08s\n- openstack-tox-py37 https://logs.opendev.org/15/670715/9/check/openstack-tox-py37/8a925a0/ : SUCCESS in 10m 50s\n- openstack-tox-docs https://logs.opendev.org/15/670715/9/check/openstack-tox-docs/52b12c9/html/ : SUCCESS in 10m 36s\n- tempest-full https://logs.opendev.org/15/670715/9/check/tempest-full/03d0826/ : SUCCESS in 1h 23m 39s\n- neutron-grenade https://logs.opendev.org/15/670715/9/check/neutron-grenade/ccf31ed/ : SUCCESS in 1h 08m 09s\n- grenade-py3 https://logs.opendev.org/15/670715/9/check/grenade-py3/67685e4/ : SUCCESS in 1h 00m 24s\n- tempest-integrated-storage https://logs.opendev.org/15/670715/9/check/tempest-integrated-storage/f549d8a/ : SUCCESS in 1h 15m 57s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/15/670715/9/check/cinder-tempest-dsvm-lvm-lio-barbican/d0714fc/ : FAILURE in 1h 28m 19s\n- cinder-tox-bandit-baseline https://logs.opendev.org/15/670715/9/check/cinder-tox-bandit-baseline/adc0a0b/ : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/15/670715/9/check/openstack-tox-functional-py36/20f9d77/ : SUCCESS in 6m 41s\n- cinder-rally-task https://logs.opendev.org/15/670715/9/check/cinder-rally-task/9e3147c/ : SUCCESS in 58m 47s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/15/670715/9/check/openstack-tox-pylint/8fb04c4/ : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/15/670715/9/check/cinder-plugin-ceph-tempest/f1126ce/ : SUCCESS in 1h 07m 05s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/15/670715/9/check/openstack-tox-functional/b3088ef/ : SUCCESS in 8m 47s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/15/670715/9/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/323ce11/ : SUCCESS in 1h 16m 30s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/15/670715/9/check/legacy-tempest-dsvm-lvm-multibackend/a8aafb0/ : SUCCESS in 1h 16m 22s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/15/670715/9/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/ce58100/ : SUCCESS in 1h 32m 28s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/15/670715/9/check/tempest-slow-py3/21a54ac/ : SUCCESS in 2h 39m 19s","accounts_in_message":[],"_revision_number":9},{"id":"7b7f1629ed2a85e9e45362fe34058903cc1843bb","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-07 02:55:37.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"63046a2cd1a3aad2086a62e0c185d41de4f27ab6","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-08-08 06:36:42.000000000","message":"Patch Set 9:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":9},{"id":"5be42ca55122302a2a54af30995b3be4adcff3c8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-08 10:26:47.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/9/check/dsvm-tempest-zadara-driver/864a528 : SUCCESS in 1h 45m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5f371f6d4b3e77df7ccb028250a88b86e2aa4f86","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-08-08 11:30:23.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e612dc642b383f31ff4b7c8ede6f61a42fb8df12","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-08 11:31:54.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"d83b9f362814f85426dfee3a21fb846fd2e0788b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-08 14:01:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/9/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c40a521 : SUCCESS in 1h 04m 38s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/9/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/acc6145 : SUCCESS in 1h 15m 13s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"ca7314ea2f778ad19e1887a73a2e3d33a3ed3c88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 16:16:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/15/670715/9/check/openstack-tox-lower-constraints/315b3cf/ : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://logs.opendev.org/15/670715/9/check/openstack-tox-pep8/1685635/ : SUCCESS in 10m 48s\n- openstack-tox-py27 https://logs.opendev.org/15/670715/9/check/openstack-tox-py27/4ccbbfb/ : SUCCESS in 9m 51s\n- openstack-tox-py36 https://logs.opendev.org/15/670715/9/check/openstack-tox-py36/f3662ef/ : SUCCESS in 8m 15s\n- openstack-tox-py37 https://logs.opendev.org/15/670715/9/check/openstack-tox-py37/b893363/ : SUCCESS in 11m 50s\n- openstack-tox-docs https://logs.opendev.org/15/670715/9/check/openstack-tox-docs/4e70521/html/ : SUCCESS in 9m 30s\n- tempest-full https://logs.opendev.org/15/670715/9/check/tempest-full/56f3824/ : SUCCESS in 1h 37m 44s\n- neutron-grenade https://logs.opendev.org/15/670715/9/check/neutron-grenade/d470b4e/ : SUCCESS in 59m 37s\n- grenade-py3 https://logs.opendev.org/15/670715/9/check/grenade-py3/6940dc1/ : SUCCESS in 1h 13m 39s\n- tempest-integrated-storage https://logs.opendev.org/15/670715/9/check/tempest-integrated-storage/8745a17/ : SUCCESS in 1h 29m 18s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/15/670715/9/check/cinder-tempest-dsvm-lvm-lio-barbican/9f698a0/ : SUCCESS in 1h 38m 25s\n- cinder-tox-bandit-baseline https://logs.opendev.org/15/670715/9/check/cinder-tox-bandit-baseline/0605447/ : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/15/670715/9/check/openstack-tox-functional-py36/04d412e/ : SUCCESS in 6m 38s\n- cinder-rally-task https://logs.opendev.org/15/670715/9/check/cinder-rally-task/c5dd88c/ : SUCCESS in 45m 28s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/15/670715/9/check/openstack-tox-pylint/a9f9d82/ : SUCCESS in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/15/670715/9/check/cinder-plugin-ceph-tempest/10d9b4f/ : SUCCESS in 1h 24m 16s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/15/670715/9/check/openstack-tox-functional/1a366b4/ : SUCCESS in 6m 06s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/15/670715/9/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/1740556/ : SUCCESS in 1h 09m 40s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/15/670715/9/check/legacy-tempest-dsvm-lvm-multibackend/d2e1150/ : SUCCESS in 1h 21m 19s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/15/670715/9/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2d008c0/ : SUCCESS in 1h 56m 59s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/15/670715/9/check/tempest-slow-py3/bfb273d/ : SUCCESS in 2h 31m 25s","accounts_in_message":[],"_revision_number":9},{"id":"14b9dccd1b6141788b7e105ed444d28a6c1ced9a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-08 17:05:25.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/15/670715/9/upstream-solidfire-check/cinder-solidfire-iSCSI/f70c5c2/ : FAILURE in 47m 28s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"62eac775c2589c16903d84a8f9d1e7bd0b512a57","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-08 17:09:52.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/9/iscsi/ : SUCCESS in 1h 04m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/9/smb/ : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":9},{"id":"074aeff4d93818336c6624e31b18afeb1ac56520","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-08 19:16:28.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/15/670715/9/check/synology-dsm-cinder-iscsi/ce624e0 : FAILURE in 36m 57s","accounts_in_message":[],"_revision_number":9},{"id":"f122418f6760f6d3d035885712a797eca4881042","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-08 19:39:58.000000000","message":"Patch Set 9:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d8d06a3 : FAILURE in 1h 31m 20s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/9/check/3par-fc-driver-master-client-pip-ssa02-dsvm/3764995 : SUCCESS in 1h 54m 18s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/15/670715/9/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/5697581 : FAILURE in 1h 26m 40s","accounts_in_message":[],"_revision_number":9},{"id":"e395612a001d97f6b1d8479bb4c3385dd762c851","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 19:43:10.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/9/check/dsvm-tempest-kvm/9376f05 : FAILURE in 37m 02s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/9/check/dsvm-tempest-exe/dab9801 : FAILURE in 36m 05s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"4d1648138ea9a79070c19f3b90600496bc76e543","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-08 19:49:05.000000000","message":"Patch Set 9:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-670715-9 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":9},{"id":"c2c09ba6885756c0379cb3d3918a8f48f1f06145","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-08 22:39:50.000000000","message":"Patch Set 9:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-15-670715-9/ns5_iscsi : FAILURE in 50m 6s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-15-670715-9/ns5_nfs : FAILURE in 53m 8s","accounts_in_message":[],"_revision_number":9},{"id":"b9f0667637ee72a2ace61042f944829cd2b408ac","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-09 01:34:41.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/15/670715/9/check/kaminario-dsvm-tempest-full-iscsi/c498c26 : SUCCESS in 1h 30m 52s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/15/670715/9/check/kaminario-dsvm-tempest-full-FC/6f4f815 : SUCCESS in 1h 27m 32s","accounts_in_message":[],"_revision_number":9},{"id":"55e8bab44b011be23d14c25a0ad1221b5ed78b34","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-09 03:20:45.000000000","message":"Patch Set 9:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/15/2481 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"b0112711f458dd35fd41f9a9a6eb777040b3d69f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-09 06:35:27.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/670715/9/check/DellEMC_SC_FC/d2f8d07/DellEMC_SC_FC/414 : SUCCESS in 1h 07m 37s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/670715/9/check/DellEMC_SC_iSCSI/3a01229/DellEMC_SC_iSCSI/396 : SUCCESS in 1h 43m 23s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"5bc435caa444e3fd008a5c7dc5f9789bb1d5b371","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 07:38:03.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/15/670715/9/check/DellEMC_VNX_iSCSI/bd3fd4b/DellEMC_VNX_iSCSI/548 : SUCCESS in 1h 45m 49s\n- DellEMC_VNX_FC http://publiclogs.emc.com/15/670715/9/check/DellEMC_VNX_FC/15dcdb3/DellEMC_VNX_FC/558 : FAILURE in 1h 10m 39s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"ef9be49fa32086af5203ca1f275b2428e2598528","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 07:41:48.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/15/670715/9/check/DellEMC_VxFlexOS/138da9a/DellEMC_VxFlexOS/536 : SUCCESS in 1h 54m 33s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":9},{"id":"cfbfcc75a60545f3fb3f1cf3322784035868955d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-09 08:15:22.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/9/check/dsvm-tempest-zadara-driver/f439fd6 : SUCCESS in 1h 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c2148c2d56fb534d224ce82f5cf3e72d7f45b585","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-09 08:41:40.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/9/check/tempest-dsvm-full-xenial/4e8d22e/ : SUCCESS in 1h 24m 19s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/9/check/devstack-plugin-ceph-dsvm-xenial/985b443/ : SUCCESS in 1h 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"439f82117fa4c5acff085e74102dd633e727f701","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-09 09:31:14.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/9/check/DellEMC_Unity_FC/c48330f/DellEMC_Unity_FC/966 : SUCCESS in 1h 06m 19s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/9/check/DellEMC_Unity_iSCSI/c2650c7/DellEMC_Unity_iSCSI/923 : SUCCESS in 1h 30m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"cbe00309839c23528a19498e3f4b4c782e8540fd","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 10:06:54.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/9/check/DellEMC_XtremIO_FC/9d2bcd5/DellEMC_XtremIO_FC/1095 : SUCCESS in 1h 34m 35s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/9/check/DellEMC_XtremIO_iSCSI/f3b34c6/DellEMC_XtremIO_iSCSI/1017 : SUCCESS in 1h 36m 15s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"ce58e5be684a2401e0ce3b6c705c0e877cd5705d","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-09 10:22:39.000000000","message":"Patch Set 9:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/670715-9/HPMSA-FC : FAILURE in 37:44.66\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/670715-9/HPMSA-iSCSI : FAILURE in 38:28.61","accounts_in_message":[],"_revision_number":9},{"id":"892e76ae312e4658e04a7d5e9f4e45ab3ac346fc","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-09 17:57:59.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"9bf7c4364ca6a786d6bf7ab71cc6ac27c39fb794","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-09 22:13:43.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/9/check-not-vote/ext-cinder-zuul/a4d5db4 : FAILURE in 2m 04s","accounts_in_message":[],"_revision_number":9},{"id":"0a7ca0da1c8c8d4e587b3523a623bd03751941bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-10 19:00:20.000000000","message":"Patch Set 9:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/355/670715/9/check/infortrend-cinder-volume-fc/ac3600dfacc2eb1bcd853841dd692060c88a68c5 : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/398/670715/9/check/infortrend-cinder-volume-iscsi/ac3600dfacc2eb1bcd853841dd692060c88a68c5 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"276a311ce0aefa119ea783c29c2f36aababb66b4","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-08-23 15:48:52.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)\n\nSeveral concerns with this patch:\n1) There are unnecessary White Space changes in the file.\n2) It appears to cover multiple unrelated changes.\n3) It changes option functionality so it really should have a release note explaining that.","accounts_in_message":[],"_revision_number":9},{"id":"4a7bd88420dc91b36b11002fb48f776ac28f5d27","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-08-25 09:00:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4072d6d0d3a4d18dc93c6b36e051ea47658084bb","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-25 09:12:28.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1667 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1703 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1870 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"e5e2b514a5caccdf83350dc357aee8251fe35260","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-25 09:45:06.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/10/check/dsvm-tempest-kvm/3c14704 : FAILURE in 33m 50s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/10/check/dsvm-tempest-exe/36a8e23 : FAILURE in 34m 25s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"bc2b2fdadc952bca282fdf86e8684ef6a43113b5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-25 09:45:16.000000000","message":"Patch Set 10:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/621/670715/10/check/infortrend-cinder-volume-iscsi/4e24811f766e2ddb0425608e3df99ec227e80a87 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/600/670715/10/check/infortrend-cinder-volume-fc/4e24811f766e2ddb0425608e3df99ec227e80a87 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"3d56f3723b43dd91e77caae0be1f58f1e4118bad","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-25 09:59:52.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/10/iscsi/ : SUCCESS in 56m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/10/smb/ : FAILURE in 58m 38s","accounts_in_message":[],"_revision_number":10},{"id":"1e28cf6273f071d47ca57f1df3f706f80806037e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-25 10:13:55.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/15/670715/10/check-cinder/Cinder-ISER-ISCSI/10e8eee : SUCCESS in 1h 01m 22s\n- Cinder-ISER-LIO http://13.74.249.42/15/670715/10/check-cinder/Cinder-ISER-LIO/ecfb5f5 : SUCCESS in 1h 11m 58s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"48d77f54ac4800c1a7fb9c5951754c8c50d69d25","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-08-25 10:14:39.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/15/670715/10/check/Cinder-AS13000/233bffb : FAILURE in 59m 15s\n- cinder-test-fc http://218.29.111.78:9000/15/670715/10/check/cinder-test-fc/57b42aa : SUCCESS in 1h 01m 08s\n- cinder-test-iscsi http://218.29.111.78:9000/15/670715/10/check/cinder-test-iscsi/4fccd72 : SUCCESS in 1h 06m 49s","accounts_in_message":[],"_revision_number":10},{"id":"60f505a4277cb59e4cbe700cc01d4f09d1bc4972","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-25 10:16:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/10/check/dsvm-tempest-infinibox-fc/1c5a674 : SUCCESS in 1h 08m 51s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/10/check/dsvm-tempest-infinibox-iscsi/c67574d : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":10},{"id":"436ade376d63fa0ed0a64c8a21ab2c8aadee8b1d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-25 10:21:24.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/10/check/DellEMC_Unity_FC/0305793/DellEMC_Unity_FC/1099 : FAILURE in 1h 10m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/10/check/DellEMC_Unity_iSCSI/0deb396/DellEMC_Unity_iSCSI/1048 : SUCCESS in 1h 20m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"03bacc8493f0e0e5b17c7d1b1ff341645007dcc0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-25 10:32:11.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"2598bbfa7e3298b67565ec47e0d785bc1c9f6404","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-25 10:39:07.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/10/check/fujitsu-eternus-dx-iscsi/6b1a2df : SUCCESS in 1h 37m 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/10/check/fujitsu-eternus-dx-fc/f707f29 : FAILURE in 1m 16s","accounts_in_message":[],"_revision_number":10},{"id":"615dd63f44fbe9be552e4be7b741f9deee6ad477","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-25 10:48:07.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/10/check/dsvm-tempest-zadara-driver/ad4b8cf : FAILURE in 1h 47m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"04c1ccfdeb267cbb3716daceb7e3e5142a075b29","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-25 10:54:12.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/10/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/8fb3c55 : SUCCESS in 1h 00m 33s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/10/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/3e667ee : FAILURE in 1h 53m 03s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"52983eb2706bb49e6bd38bb805eb2462b5e006ef","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-25 10:55:53.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/10/check/NEC-ISCSI/b8f74c7 : FAILURE in 1h 52m 45s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/10/check/NEC-FC/d2c4f3b : FAILURE in 1h 44m 59s","accounts_in_message":[],"_revision_number":10},{"id":"dbee9594dec519c53d1b93c04c48e60fc0078a61","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-25 11:03:08.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/10/check/DellEMC_XtremIO_FC/642d3d7/DellEMC_XtremIO_FC/1239 : SUCCESS in 1h 28m 04s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/10/check/DellEMC_XtremIO_iSCSI/af4d498/DellEMC_XtremIO_iSCSI/1160 : SUCCESS in 1h 34m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"3b03b349d6409394e6e895ea9cf0e05ceebbaf90","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-25 11:22:25.000000000","message":"Patch Set 10:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/622/670715/10/check/infortrend-cinder-volume-iscsi/4e24811f766e2ddb0425608e3df99ec227e80a87 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/601/670715/10/check/infortrend-cinder-volume-fc/4e24811f766e2ddb0425608e3df99ec227e80a87 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"b67e9d7f645e0fdea52a56619fbe0d3a4ce3ce84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 11:22:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a940fe60c974a9d9c07af9daaa24615 : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5bc707f1ab4acda41af92260754adf : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6389daab36b04aa3bd8c93ba054d0bc0 : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/059f0156ac30499e9f08379085694b3e : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7298f876fae411db6b0934a5202425b : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3afe96f1472a48f084316cddea944000 : SUCCESS in 7m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3b5dcfe37ffa46b39f31f99b7c64e94f : SUCCESS in 1h 48m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/500e1b90f190403dae8ed795b6996ed9 : SUCCESS in 1h 00m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/555fc345fb474807b50aaa7c677da238 : SUCCESS in 58m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84dce181f502415d9dc394d5ac30b03a : SUCCESS in 1h 04m 44s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6279b2f3880746a28fb70d8a9d1ae753 : SUCCESS in 1h 21m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/48c95919572945ca9eb3b0c3091ddcd9 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44ad3723cd584c2a90183b983b51c88a : SUCCESS in 5m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2c5d58bf0e54f6fa6f660e0c3f2b712 : SUCCESS in 35m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47302b9a41d34450b966f49055c651a9 : SUCCESS in 4m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd179aa602214ae191f47128369648c5 : SUCCESS in 1h 15m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8392ab905c4b468dbed97196fd6a1e25 : SUCCESS in 5m 01s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38e8cf5bc9d8499ba958ac5972d972b1 : SUCCESS in 1h 07m 33s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f9266e9d637c4cfeaccc81c9cef35f36 : SUCCESS in 1h 25m 20s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/80871c1a722f4db1a77e06ec9a5f1ae0 : SUCCESS in 1h 37m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c46ab4e492834aa3a987824465415457 : SUCCESS in 2h 16m 21s","accounts_in_message":[],"_revision_number":10},{"id":"5e5ac3b79d31ebd474f88df9c2cd93d677f08f63","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-25 12:01:35.000000000","message":"Patch Set 10:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-15-670715-10/ns5_iscsi : SUCCESS in 36m 35s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-15-670715-10/ns5_nfs : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":10},{"id":"801dc9fa050486fafd22431ca7acd6288b750ad6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-25 12:20:04.000000000","message":"Patch Set 10:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-670715-10 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":10},{"id":"57e794c19b19fb842030af50a5e196f5a04cd5b2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-25 12:22:14.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/15/670715/10/upstream-check/cinder-cDOT-iSCSI/0b08e33/ : SUCCESS in 56m 06s\n- cinder-cDOT-NFS http://13.56.159.105/logs/15/670715/10/upstream-check/cinder-cDOT-NFS/988714f/ : SUCCESS in 51m 23s\n- cinder-cDOT-FCP http://13.56.159.105/logs/15/670715/10/upstream-check/cinder-cDOT-FCP/bdb307c/ : SUCCESS in 58m 45s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"59be6137baf00b6759cd69a6dee512a88fb04680","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-25 12:25:52.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/15/670715/10/upstream-solidfire-check/cinder-solidfire-iSCSI/90d6738/ : SUCCESS in 1h 01m 55s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"f56971216affa413beed51a3e641aa57b853ec88","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-25 12:29:37.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/15/670715/10/check/synology-dsm-cinder-iscsi/2afa5dc : FAILURE in 1h 06m 20s","accounts_in_message":[],"_revision_number":10},{"id":"26cf2d67266b46ce88b280e01e3267bc145dca1e","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-25 12:45:23.000000000","message":"Patch Set 10:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/15/2586 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"e6451da8e23b549d6ca0423bce712a4bcc93e581","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-25 12:56:20.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/10/check/DellEMC_Unity_FC/3ec11ff/DellEMC_Unity_FC/1100 : SUCCESS in 1h 03m 11s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/10/check/DellEMC_Unity_iSCSI/7f88bf4/DellEMC_Unity_iSCSI/1049 : SUCCESS in 1h 33m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"6333a98991b48825004a0e25a4754dc921ed68c3","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-25 12:58:53.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/15/670715/10/check/kaminario-dsvm-tempest-full-iscsi/45f06e8 : FAILURE in 5m 36s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/15/670715/10/check/kaminario-dsvm-tempest-full-FC/8ed19e6 : SUCCESS in 1h 35m 44s","accounts_in_message":[],"_revision_number":10},{"id":"f0fa43a42a1a19ef99865bd64c5a0d5f52129e43","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-25 13:00:32.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/10/check/DellEMC_XtremIO_FC/49eeaec/DellEMC_XtremIO_FC/1240 : SUCCESS in 1h 28m 24s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/10/check/DellEMC_XtremIO_iSCSI/4905a01/DellEMC_XtremIO_iSCSI/1161 : SUCCESS in 1h 37m 34s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"c34745ccd6db5452056b877d9891eba5acfdf25b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-08-25 13:15:22.000000000","message":"Patch Set 10:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/670715-10/Lenovo-FC : FAILURE in 39:04.75\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/670715-10/Lenovo-iSCSI : FAILURE in 41:08.36","accounts_in_message":[],"_revision_number":10},{"id":"17e999dc707e7bf3bac07451b77231301808e86c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-25 13:16:35.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/15/670715/10/check/DellEMC_VxFlexOS/353e111/DellEMC_VxFlexOS/614 : SUCCESS in 1h 53m 34s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":10},{"id":"4f6e76262d74948880df19f2c778f606b0900a0f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-25 13:20:16.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/670715/10/check/DellEMC_SC_FC/9f77f53/DellEMC_SC_FC/495 : SUCCESS in 1h 04m 52s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/670715/10/check/DellEMC_SC_iSCSI/cb79908/DellEMC_SC_iSCSI/474 : SUCCESS in 1h 57m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"7163cdf132e67f912ed83186fe1e1d996006e974","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-25 13:21:24.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"43c3a4a53d05ad65c5489f7d467b8994b1bd5bcd","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-25 13:23:47.000000000","message":"Patch Set 10:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/670715-10/HPMSA-FC : FAILURE in 43:23.44\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/670715-10/HPMSA-iSCSI : FAILURE in 43:18.71","accounts_in_message":[],"_revision_number":10},{"id":"ae26b00713fc49ddfb7556bee8fe936706fd4305","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-08-25 13:29:00.000000000","message":"Patch Set 10:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/670715-10/Seagate-FC : SUCCESS in 43:48.89\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/670715-10/Seagate-iSCSI : SUCCESS in 46:54.04","accounts_in_message":[],"_revision_number":10},{"id":"978b14d2b6a59c62b4884f699062526bbc5af9b8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-25 13:35:29.000000000","message":"Patch Set 10:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a4e692e : SUCCESS in 2h 05m 06s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/10/check/3par-fc-driver-master-client-pip-ssa02-dsvm/c8cecb8 : SUCCESS in 2h 10m 13s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/15/670715/10/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/4cc2704 : FAILURE in 1h 29m 11s","accounts_in_message":[],"_revision_number":10},{"id":"a51b6c19520425ceb39dd0a463db992f2e582804","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-25 14:16:55.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/15/670715/10/check/DellEMC_VNX_iSCSI/48368ce/DellEMC_VNX_iSCSI/631 : FAILURE in 2h 53m 55s\n- DellEMC_VNX_FC http://publiclogs.emc.com/15/670715/10/check/DellEMC_VNX_FC/85343a2/DellEMC_VNX_FC/644 : FAILURE in 1h 00m 32s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"4c56a4cf39eb7c544a6aae88e52b06354a0a31f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-25 14:27:15.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/10/check/tempest-dsvm-full-xenial/65795c3/ : SUCCESS in 1h 35m 47s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/10/check/devstack-plugin-ceph-dsvm-xenial/8ef20ef/ : SUCCESS in 1h 38m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9efdb53b53c9d7e12a5186cc04f92cbdd9df1517","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-08-26 18:24:37.000000000","message":"Patch Set 10:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":10},{"id":"351e79f9f82fb6dd8651afba6aa3b78e33dbfd7c","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-08-26 19:18:10.000000000","message":"Patch Set 10:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":10},{"id":"5063f8cba54c793db2fa20de128e12f020ce8e27","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-26 21:03:27.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/10/check/dsvm-tempest-zadara-driver/efb5f72 : FAILURE in 1h 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b6048272441e4708058a5ceb156901eddee11619","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-08-27 02:41:18.000000000","message":"Patch Set 10:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":10},{"id":"c36bbea5124ea9f6b4e693c2a57555acbc2d1fad","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-27 04:26:11.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/10/check/dsvm-tempest-zadara-driver/4f7072a : SUCCESS in 1h 44m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"18d7f3338fe66700f9f8b36e8c8631d962ec7311","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-08-27 14:29:21.000000000","message":"Patch Set 10: Code-Review-1\n\nNeed to add a release note about this.","accounts_in_message":[],"_revision_number":10},{"id":"516dd18534a1f9bb4c534bc8f15f0bccb76f7e43","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-08-28 09:36:17.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2357b2587ef86cfe0a37f2818fe3dd24c30d4db4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-28 09:36:48.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"0a46adc45c18a9ead1dfc61ac3f8ddb14412784f","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-28 09:41:58.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1894 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1702 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1727 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"8a0b140c206447abbd526bbc5275093f6374d2fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-28 10:20:49.000000000","message":"Patch Set 11:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/675/670715/11/check/infortrend-cinder-volume-iscsi/fd7ef9a8348e9b345f329fa203e9fee3af5b2450 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/653/670715/11/check/infortrend-cinder-volume-fc/fd7ef9a8348e9b345f329fa203e9fee3af5b2450 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"d4218381a6379f45bc81a895055d677d26930d70","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 10:23:43.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/11/check/dsvm-tempest-kvm/3d61202 : FAILURE in 35m 49s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/11/check/dsvm-tempest-exe/853b172 : FAILURE in 35m 33s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"3a4dadd16e381e3f7b817062fe24b8ee1f0250ec","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-28 10:50:02.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/11/check/dsvm-tempest-infinibox-fc/87bda46 : SUCCESS in 53m 49s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/11/check/dsvm-tempest-infinibox-iscsi/99236de : SUCCESS in 1h 13m 05s","accounts_in_message":[],"_revision_number":11},{"id":"986e920d9fe3652af2009f9de9413a2b6854d548","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-28 10:50:41.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/11/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/6bb72ce : SUCCESS in 1h 06m 57s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/11/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/c20e1de : SUCCESS in 1h 14m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":11},{"id":"7b9bd515c02f2b7f2859b191a3acf689894631ea","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-28 10:50:46.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/11/iscsi/ : SUCCESS in 1h 13m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/11/smb/ : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":11},{"id":"be869291ed4379dcf869970fc858ec37480e62a8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-28 11:11:23.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/11 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"33cbd48db42aecdd398f2a3e3b2cd067822a21ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 12:12:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/319d0763f5244201a7070f7f26500212 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e698b8ff56dc4e4f8995abf4798a50af : SUCCESS in 8m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/585c6f78a0fc4151a2f730e3f51ba0fe : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc768f55058d45d3adb120b0fb3a1dc6 : SUCCESS in 9m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2effacdc2448472ea3e337957f12fd72 : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a86897e584426a97324b32cc354e1d : SUCCESS in 12m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/787e70ac539644978bab826d585e8adc : SUCCESS in 1h 57m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/409185e402874958bba9b584ee8bf059 : SUCCESS in 1h 03m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/efd8a9eefd7f41bd97af38c90511795d : SUCCESS in 1h 03m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d939575d5e5848e2aac284c9abe3b598 : SUCCESS in 1h 27m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41ee569e06f344d6a0ec6d9d519b58cb : SUCCESS in 20m 19s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5aa97cb3ad9a41b996d358bda75d587d : SUCCESS in 1h 36m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d5d7520122394fb8867888f8e94e2725 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d102f1bf8f674e93bbb6270d0f43d781 : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7dcfedec1b974dd8a056082e275acba3 : SUCCESS in 47m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b877905e3734bb8a5a496c4a2ba6929 : SUCCESS in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/08abd4fcef4c4deabb83942d89ec6b73 : SUCCESS in 1h 14m 40s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/749daf9684b441e298b8b141f77468c9 : SUCCESS in 5m 58s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc8ad147e7fa4792b7ca80e2c5ac4a9f : SUCCESS in 1h 07m 38s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d2f78af3a08943da88dc2ee6de9b6982 : FAILURE in 1h 27m 09s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/1bb7aaae542c4155b25672c894688bb0 : SUCCESS in 1h 36m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a63c0d0f13b4484bbf0b4838364f7148 : SUCCESS in 2h 15m 04s","accounts_in_message":[],"_revision_number":11},{"id":"af1e45a0e23de4b2db704bf4ec2b1b95558ec705","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 12:36:56.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/11/check/NEC-ISCSI/4798846 : FAILURE in 1h 41m 00s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/11/check/NEC-FC/487c68d : FAILURE in 1h 53m 01s","accounts_in_message":[],"_revision_number":11},{"id":"1ca507be9cb1f81986103e2d77d1d07067a4e328","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-28 12:40:10.000000000","message":"Patch Set 11:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/11 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"62b494d7328f38406e8c8bdb32d5cf899008c1a2","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-28 12:48:34.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/11/check/dsvm-tempest-zadara-driver/6ff7767 : SUCCESS in 1h 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f200310f5616423c0ae683ffb37106e15adeb8f7","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-28 12:53:51.000000000","message":"Patch Set 11:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/15/2618 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"da06d517161c03434e818bd48b2d0d1c1af57a56","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-28 12:54:59.000000000","message":"Patch Set 11:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-15-670715-11/ns5_iscsi : SUCCESS in 39m 4s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-15-670715-11/ns5_nfs : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":11},{"id":"53bf4293b9e5e8af66b097fa8d2e60249be32f36","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-28 13:09:11.000000000","message":"Patch Set 11:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-670715-11 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":11},{"id":"036cdeb39f7d9278f882ae327f67af1ad22505d6","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-28 13:24:24.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/15/670715/11/check/synology-dsm-cinder-iscsi/ba6bbaa : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":11},{"id":"9da5270c3e81ddbcca5e0e11cc147f828da3fd96","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 13:49:18.000000000","message":"Patch Set 11:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-15-670715-11-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"30edf921aea43c980140f658aba96a9f51614e0a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-28 14:24:02.000000000","message":"Patch Set 11:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4933816 : FAILURE in 2h 08m 19s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/11/check/3par-fc-driver-master-client-pip-ssa02-dsvm/5213851 : SUCCESS in 2h 07m 57s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/15/670715/11/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/b0b1230 : FAILURE in 1h 31m 52s","accounts_in_message":[],"_revision_number":11},{"id":"c2f35d3c6d0ce849fb640267d59d4b25ff838a1b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-28 14:28:53.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/670715/11/check/DellEMC_SC_FC/9b8bbaa/DellEMC_SC_FC/527 : FAILURE in 1h 05m 20s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/670715/11/check/DellEMC_SC_iSCSI/b8e0063/DellEMC_SC_iSCSI/504 : FAILURE in 1h 37m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":11},{"id":"84c9d4d780108dd52f1d5c954cbf436f382fc27d","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-28 14:48:51.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/15/670715/11/check/kaminario-dsvm-tempest-full-iscsi/f419e0f : SUCCESS in 1h 41m 51s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/15/670715/11/check/kaminario-dsvm-tempest-full-FC/e7a2aac : SUCCESS in 1h 37m 54s","accounts_in_message":[],"_revision_number":11},{"id":"659dced794f3039aa6e067c2b3b216f8e8b6493f","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-08-28 15:12:32.000000000","message":"Patch Set 11:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/670715-11/Seagate-FC : SUCCESS in 42:31.76\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/670715-11/Seagate-iSCSI : SUCCESS in 44:31.72","accounts_in_message":[],"_revision_number":11},{"id":"c41224862fbf543b038b0cacf01bc1244c7ea2b4","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-28 15:38:45.000000000","message":"Patch Set 11:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/670715-11/HPMSA-FC : FAILURE in 43:12.82\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/670715-11/HPMSA-iSCSI : FAILURE in 45:35.53","accounts_in_message":[],"_revision_number":11},{"id":"0ba9351f5fb3e52464a4173d53029dda77835ee2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-28 16:29:23.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/11/check/DellEMC_Unity_FC/e7164ce/DellEMC_Unity_FC/1143 : SUCCESS in 1h 04m 37s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/11/check/DellEMC_Unity_iSCSI/69c23cf/DellEMC_Unity_iSCSI/1092 : SUCCESS in 1h 26m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"ac5ef73bff4f40034d30d339eb8801f991183b65","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 17:30:18.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/15/670715/11/check/DellEMC_VxFlexOS/751eb34/DellEMC_VxFlexOS/645 : SUCCESS in 1h 56m 37s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":11},{"id":"43527b4efadad6d7a1a3da474dd4e2a8ebd1436c","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-08-28 18:26:13.000000000","message":"Patch Set 11:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/670715-11/Lenovo-FC : FAILURE in 44:04.10\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/670715-11/Lenovo-iSCSI : FAILURE in 44:18.47","accounts_in_message":[],"_revision_number":11},{"id":"ade3107fa50e7152475449548ecfe8f723777f33","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 19:06:26.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/15/670715/11/check/DellEMC_VNX_iSCSI/7ad7f6c/DellEMC_VNX_iSCSI/662 : SUCCESS in 1h 47m 30s\n- DellEMC_VNX_FC http://publiclogs.emc.com/15/670715/11/check/DellEMC_VNX_FC/2133b81/DellEMC_VNX_FC/677 : FAILURE in 1h 26m 33s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":11},{"id":"32602b263829751d92fe8b83cfefc80deea577f1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 23:29:36.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/11/check/DellEMC_XtremIO_FC/3f6337d/DellEMC_XtremIO_FC/1282 : SUCCESS in 1h 27m 32s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/11/check/DellEMC_XtremIO_iSCSI/3799014/DellEMC_XtremIO_iSCSI/1203 : SUCCESS in 2h 00m 08s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":11},{"id":"8dbf26ec5fcda3a9e4967c3c8d88168c67ec6200","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-09-03 16:40:34.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"20cc7615a028fa3279af0d9f52ee388d696fec59","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-09-05 06:36:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1d941bf81c2c965ede13a63adcf399836563eb15","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2019-09-05 07:05:18.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"09f1271f21ff62c023d566849fa760f55fe369ce","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 07:13:20.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/12/check/dsvm-tempest-kvm/f019fa2 : FAILURE in 36m 28s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/12/check/dsvm-tempest-exe/7a4fc51 : FAILURE in 35m 41s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"ba6eee55b0ec05e1c7e43181bf70c83d1aa66dc8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-09-05 07:52:01.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":12},{"id":"e96fc1dc5ab5bb54831b164dfd363987719cba15","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-09-05 07:55:22.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/12/check/dsvm-tempest-infinibox-fc/54c1dac : FAILURE in 21m 25s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/12/check/dsvm-tempest-infinibox-iscsi/763f6e2 : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":12},{"id":"573200379affc3571087450390ff50f3f842cfca","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-05 08:06:10.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/12/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/cd3cc4f : SUCCESS in 1h 29m 41s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/12/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a81692f : FAILURE in 30m 18s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":12},{"id":"6f555d082351fa09ce55ef5398a153771df06ec0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-09-05 08:15:15.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"8afdac9b8e06b30f8c4ba1ea3df8d45810a50747","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-09-05 08:41:26.000000000","message":"Patch Set 12:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/670715/12/check/fujitsu-eternus-dx-iscsi/8a08908 : SUCCESS in 1h 57m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/670715/12/check/fujitsu-eternus-dx-fc/d65761e : SUCCESS in 2h 03m 17s","accounts_in_message":[],"_revision_number":12},{"id":"ccf46b95de8f8d92f18c859b9e8cfa7975d0bcbc","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-05 09:19:17.000000000","message":"Patch Set 12:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/15/670715/12/check-cinder/Cinder-ISER-ISCSI/537914c : FAILURE in 1s\n- Cinder-ISER-LIO http://13.74.249.42/15/670715/12/check-cinder/Cinder-ISER-LIO/b640d53 : SUCCESS in 1h 06m 57s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"2c28989edb14c2169cb2359cb97103809377e11b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-09-05 10:43:19.000000000","message":"Patch Set 12:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/818/670715/12/check/infortrend-cinder-volume-iscsi/25a920b4860d1c2cfc72d1f401c09fa1f91a0590 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/796/670715/12/check/infortrend-cinder-volume-fc/25a920b4860d1c2cfc72d1f401c09fa1f91a0590 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"a66382fc9fc5924b584ec07a8a3f9fcd3657a944","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-09-05 11:31:44.000000000","message":"Patch Set 12:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"1b028a8186bc95ad1e299f7fedd9b2b108fbf460","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 14:37:39.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/12/check/NEC-ISCSI/ae57ea4 : FAILURE in 34m 25s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/12/check/NEC-FC/b2df40e : FAILURE in 37m 25s","accounts_in_message":[],"_revision_number":12},{"id":"ae9330a97acac85ab0f0cb97311db084984c87a8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-09-05 18:06:23.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/15/670715/12/check/cinder-storpool-tempest/57b655b/ : SUCCESS in 1h 27m 45s","accounts_in_message":[],"_revision_number":12},{"id":"ed3374ff6f8ca60ac6744fc572d4a9882a0f5390","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 01:54:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a447abb53c064aa3ba12d6bc3b2dc416 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b1dfeee4ab2479f9c72f8a8badb2167 : SUCCESS in 8m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f7524d487154dda8e4982571173ff5f : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cea22a54b979409da5fab4cc28fbbe31 : SUCCESS in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6f7632d3efb468390bd49a5f43e4885 : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b54188f70a4beca5f9e5f274eeaab3 : SUCCESS in 12m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0d6bd398a3e04042a0fc9e52bddbb4f5 : SUCCESS in 1h 44m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d9bf7fd2e65a4ad58a78516a2b76b36f : SUCCESS in 1h 01m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f077a1991bec40ea97dadebb44fa60a9 : SUCCESS in 1h 28m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b83ae58a64ac4322b2fb5b26cf91119d : SUCCESS in 1h 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36ef8ad000e846508c003f1431353212 : SUCCESS in 14m 18s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b91576cddb6740c38f91f14537c7ca2b : FAILURE in 2h 16m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/065eea297d844b088d4f1969b1bee91b : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/763c5344befd4820838bc69f564abc7e : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/74259b7d0e1e49cfb059da9fffa322b6 : SUCCESS in 42m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e3a546c145584f4f8fe5887b84d73e3c : SUCCESS in 4m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a1aab3eba0349f39020653fda435403 : SUCCESS in 1h 21m 29s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/37df627858af4224b4068a6a998826ff : SUCCESS in 4m 32s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/92092fb6ef7f446fb8060f224eb52d2f : SUCCESS in 1h 19m 32s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/43c467f23b724bda8a443290569f8cc5 : SUCCESS in 1h 36m 01s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/99414e266a1c41ec882b4e6be1f3026c : SUCCESS in 1h 36m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6217685d8fc46958d272a4cbf0eec8e : SUCCESS in 2h 19m 40s","accounts_in_message":[],"_revision_number":12},{"id":"a12217f55db566a96faf99317b996c433d325fe8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-06 07:37:45.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/12/check-not-vote/ext-cinder-zuul/b3ea947 : FAILURE in 23m 47s","accounts_in_message":[],"_revision_number":12},{"id":"7a5223345bd166f8698cde923729a7eb7123c6a5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 08:57:08.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/12/check/DellEMC_XtremIO_FC/0a458c2/DellEMC_XtremIO_FC/1358 : SUCCESS in 1h 33m 44s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/12/check/DellEMC_XtremIO_iSCSI/38ed860/DellEMC_XtremIO_iSCSI/1272 : FAILURE in 1h 37m 42s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"14bd3d3fc95ed15ed45bb4e2d8f47fccd2c46b79","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-09-07 16:32:48.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"7db05dd9e27b32b72c044b49669a77eae5e56e33","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2019-09-07 16:57:43.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"1d805fb83da61f3071e026812761531f3991a353","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-09-07 17:15:16.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/12/check/dsvm-tempest-kvm/10c0788 : FAILURE in 36m 23s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/12/check/dsvm-tempest-exe/6eeebf6 : FAILURE in 41m 11s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"8c4ab53a67376175e03dd1c50ff2e5a39c4bc7f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-09-07 17:33:42.000000000","message":"Patch Set 12:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/868/670715/12/check/infortrend-cinder-volume-iscsi/25a920b4860d1c2cfc72d1f401c09fa1f91a0590 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/846/670715/12/check/infortrend-cinder-volume-fc/25a920b4860d1c2cfc72d1f401c09fa1f91a0590 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"5f1e57377bb45aac8c303db97fe064dd1aa0c889","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-07 17:42:24.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/12/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/24ebeb3 : SUCCESS in 1h 00m 58s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/12/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/7a35a03 : SUCCESS in 1h 09m 18s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":12},{"id":"7396d8b11a7c5508f26752234170b4ac68f344f5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-09-07 18:00:06.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/12/iscsi/ : FAILURE in 1h 26m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/12/smb/ : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":12},{"id":"a19c43cdd739df18619f3241298426ad8a703210","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-09-07 18:02:09.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/15/670715/12/check/cinder-storpool-tempest/48106ad/ : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":12},{"id":"15b4d7ab3a523ec061973f928519f0b47491d570","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-07 18:19:08.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/12/check/tempest-dsvm-full-xenial/b9d173b/ : SUCCESS in 1h 45m 51s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/12/check/devstack-plugin-ceph-dsvm-xenial/fc82b4b/ : SUCCESS in 1h 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"cbafa105e60285c6d58388eab4e5107bb5f753c7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-09-07 18:29:52.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/670715/12/check/dsvm-tempest-cisco-zonemanager-job/6ac3336/ : SUCCESS in 1h 46m 19s","accounts_in_message":[],"_revision_number":12},{"id":"ae3b6407ae7c7b64443a4ce780c949208c2e9793","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 19:05:19.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b02774782504ec8b1c53f19863e7241 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15be738c69594b868d2a983d94f2434f : SUCCESS in 9m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f1b87ed3bb14a509862c6375a82a754 : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4206ea07a5ab41b384f1e5acdf713724 : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a78ec24eb3d9477ab1bb37c9c745a843 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17bb18e1d29a4247b8b840e1bb962fb4 : SUCCESS in 12m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d9757dca05f04feda320ae56c6698ef7 : SUCCESS in 1h 31m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/926f61265b4a4939ba144d63a3962c72 : SUCCESS in 57m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f72db9f33ddd43d4a4e461a48b854dc3 : SUCCESS in 58m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/356ce11e109e489ea04ecbefbe9abc6d : SUCCESS in 1h 19m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9319efe0dfeb46cbb9396d36a39cadce : SUCCESS in 17m 52s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2a7a5d182f834d689dcd59fbef759d77 : SUCCESS in 1h 27m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/477e16654d18416e944ccea9fa20a1f2 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6aadcd5da674d26b03fe454af6319b1 : SUCCESS in 4m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c1553d4c8d284005803e24b84eb95c40 : SUCCESS in 40m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3cb8ecae5d7b4df0b2e5a0306303e61b : SUCCESS in 3m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3decf5fee35a4ccaa3dc13eb2ed6072a : SUCCESS in 1h 12m 26s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b6a4111707c24ee49c0b9e570bfe77f4 : SUCCESS in 4m 32s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3d920036fcd84206bc67617ef6243537 : SUCCESS in 1h 13m 01s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5b31072179ca400ebc035301348e503f : SUCCESS in 1h 19m 44s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/77c13aaf5edd4c23bb66716a81d27084 : SUCCESS in 1h 35m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f48aaf10e46043f2b0cb30a7ceb99065 : SUCCESS in 2h 26m 02s","accounts_in_message":[],"_revision_number":12},{"id":"aabbe4b4a675028b5910e8079fe13a5b73dc82ca","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-09-07 19:45:02.000000000","message":"Patch Set 12:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-15-670715-12/ns5_iscsi : SUCCESS in 39m 30s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-15-670715-12/ns5_nfs : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":12},{"id":"0df3e102fe4a439baffb098375fe91b2149d40c3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-07 20:02:02.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/12/check-not-vote/ext-cinder-zuul/532e2d3 : FAILURE in 23m 09s","accounts_in_message":[],"_revision_number":12},{"id":"84b3eca4dce5b1550ebcd1ee699caeaa57016fbd","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-09-07 20:08:04.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/15/670715/12/check/synology-dsm-cinder-iscsi/10a2a7f : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":12},{"id":"4adeb20702e8b68a55df58af73f7a9c95d926538","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-07 20:12:31.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/15/670715/12/upstream-check/cinder-cDOT-iSCSI/efb5621/ : SUCCESS in 51m 27s\n- cinder-cDOT-NFS http://13.56.159.105/logs/15/670715/12/upstream-check/cinder-cDOT-NFS/ed18d35/ : SUCCESS in 52m 24s\n- cinder-cDOT-FCP http://13.56.159.105/logs/15/670715/12/upstream-check/cinder-cDOT-FCP/498d7e3/ : SUCCESS in 1h 06m 42s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"0197a8b6cd276f35e0d942102d7a310839542e4c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-07 20:13:33.000000000","message":"Patch Set 12:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-670715-12 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":12},{"id":"1ae762618fcdcfea6365c28ec124cce90366e36c","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-09-07 20:23:24.000000000","message":"Patch Set 12:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-15-670715-12-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"6fa60ac6edefc60a83c5ece5633309307b1eccf6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-09-07 20:24:32.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/15/670715/12/upstream-solidfire-check/cinder-solidfire-iSCSI/d631b5d/ : SUCCESS in 1h 18m 14s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"fd2f31924fff1aa58d6ab6573afd646e1b656432","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-09-07 20:33:29.000000000","message":"Patch Set 12:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/69fdaee : FAILURE in 1h 24m 39s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/12/check/3par-fc-driver-master-client-pip-ssa02-dsvm/6bdc3ac : FAILURE in 1h 25m 38s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/15/670715/12/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/26b47ee : FAILURE in 33m 04s","accounts_in_message":[],"_revision_number":12},{"id":"1ff2965b65113dfa05c6691ba510e91797b12e33","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-09-07 20:51:36.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/15/670715/12/check/kaminario-dsvm-tempest-full-iscsi/79c01e8 : SUCCESS in 1h 45m 52s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/15/670715/12/check/kaminario-dsvm-tempest-full-FC/31732ec : SUCCESS in 1h 39m 18s","accounts_in_message":[],"_revision_number":12},{"id":"05835052f71325cbce348eec4a5f20d3eeeb0f26","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-09-07 20:52:48.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/670715/12/check/DellEMC_SC_FC/bce01fc/DellEMC_SC_FC/613 : SUCCESS in 1h 18m 10s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/670715/12/check/DellEMC_SC_iSCSI/79adb46/DellEMC_SC_iSCSI/587 : SUCCESS in 1h 47m 13s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":12},{"id":"d4d0c270d56b220e38b34f6b878889cc6c223e02","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-07 21:04:59.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/15/670715/12/check/DellEMC_VxFlexOS/4a33d27/DellEMC_VxFlexOS/728 : SUCCESS in 1h 59m 23s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":12},{"id":"3518d423cd779ffc417e92173f646700a92353ee","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-09-07 22:14:44.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/12/check/dsvm-tempest-zadara-driver/de1e1bc : FAILURE in 1h 53m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4a7dd648414436b3ce4b84c60a8f6074d7225291","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-09-07 22:55:41.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/15/670715/12/check/DellEMC_VNX_iSCSI/0e6b7e3/DellEMC_VNX_iSCSI/744 : SUCCESS in 1h 57m 10s\n- DellEMC_VNX_FC http://publiclogs.emc.com/15/670715/12/check/DellEMC_VNX_FC/bde7cbf/DellEMC_VNX_FC/765 : FAILURE in 1h 09m 17s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":12},{"id":"8b0c6d56725a38c3bd615b20edcdd773e23cd920","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-09-07 23:52:33.000000000","message":"Patch Set 12:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/670715-12/Seagate-FC : SUCCESS in 39:20.34\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/670715-12/Seagate-iSCSI : SUCCESS in 44:09.66","accounts_in_message":[],"_revision_number":12},{"id":"0e9d2f8838a565f877077caee6226d6c4d3d9e85","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-09-08 00:54:01.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/12/check/DellEMC_Unity_FC/10a75e6/DellEMC_Unity_FC/1255 : SUCCESS in 1h 10m 16s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/12/check/DellEMC_Unity_iSCSI/0a87908/DellEMC_Unity_iSCSI/1201 : SUCCESS in 1h 22m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"1d7ba24ebbb96f65cf44d29e397632f81215ac4c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-09-08 04:35:21.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/12/check/DellEMC_XtremIO_FC/849fc0d/DellEMC_XtremIO_FC/1384 : SUCCESS in 1h 30m 54s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/12/check/DellEMC_XtremIO_iSCSI/f479685/DellEMC_XtremIO_iSCSI/1298 : FAILURE in 1h 31m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"3d4d9f94373628dc6499fad815740d044fd34287","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-09-08 05:15:50.000000000","message":"Patch Set 12:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/670715-12/HPMSA-FC : SUCCESS in 47:03.28\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/670715-12/HPMSA-iSCSI : SUCCESS in 52:00.99","accounts_in_message":[],"_revision_number":12},{"id":"fc324834c02329cb7037404f508e941193b3fdfb","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-09-08 05:48:26.000000000","message":"Patch Set 12:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/670715-12/Lenovo-FC : SUCCESS in 46:38.64\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/670715-12/Lenovo-iSCSI : FAILURE in 45:29.15","accounts_in_message":[],"_revision_number":12},{"id":"e5b140eacf31c1cae7230ce52b774381eba20305","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-09-08 05:55:12.000000000","message":"Patch Set 12:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"e3b7fe12d0fd667d77a60e14fca6d55b0cc85fc3","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-09-08 11:47:14.000000000","message":"Patch Set 12:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":12},{"id":"1d77509b5c20d6ad38fb15f8c717df873818ed9a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-09-08 12:31:07.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/12/check/dsvm-tempest-zadara-driver/7982a6a : ABORTED in 43m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"96d1cb24bd941cd83a7b0dd4ff7d90712c16d94b","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-09-08 15:10:03.000000000","message":"Patch Set 12:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/15/2714 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"30945ec474679387152a907a59901869cb4dff79","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-08 18:48:28.000000000","message":"Patch Set 12:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":12},{"id":"c3b168c1709a507d8122c879c4c2184f2198a441","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-09-09 01:51:08.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/12/check/dsvm-tempest-zadara-driver/2a8d31d : SUCCESS in 1h 42m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"41fd07da7da82cd6221980105516d7ce8e441edc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-10 13:58:59.000000000","message":"Patch Set 12: Code-Review-1\n\n(6 comments)\n\nConfigs are deprecated, but not actually deprecated. That needs to be updated to handle things correctly.","accounts_in_message":[],"_revision_number":12},{"id":"460d432c54665fa3d51cab9d313c0fd69b7f70bb","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-09-10 15:47:10.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"e94d69b9d3e08a112c48d029fdb513b0cfcd2871","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-09-10 15:47:42.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"bccc95d71387d24d12c68fa73f978bd9123cf720","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-09-10 15:51:02.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"17a8307af7facfd8425bf029683286cf3fe96fd5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-10 16:00:59.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b0d998d5021fd560a766ab8bcff70bed5dd956bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-09-10 16:13:50.000000000","message":"Patch Set 13:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/915/670715/13/check/infortrend-cinder-volume-iscsi/75b272548abb6829bfec0637396aaab4a1d4ca8d : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/893/670715/13/check/infortrend-cinder-volume-fc/75b272548abb6829bfec0637396aaab4a1d4ca8d : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"a1ece4498f021fe0364190c7ba7805de9ed93ea3","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-09-10 16:29:19.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"5bf63a11e8ddc30487a72f5321f878eeff40d0a5","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-09-10 16:29:52.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"6e113ae927dd77f5699862575341ffe15905e873","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-09-10 16:31:05.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"0d40b3c6c9a4e29599ed2e25d3a2bb205ccda048","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-09-10 16:50:46.000000000","message":"Patch Set 14:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/916/670715/14/check/infortrend-cinder-volume-iscsi/39c1a0fd803a81435519393901767c6c9b624caf : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/894/670715/14/check/infortrend-cinder-volume-fc/39c1a0fd803a81435519393901767c6c9b624caf : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"51153db3b20bfea6feed1d10ee9772b3bf68dadf","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-09-10 17:39:16.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/14/iscsi/ : SUCCESS in 1h 09m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/14/smb/ : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":14},{"id":"44d5faf712d8a300bdaa575b378cfb1b913c0f27","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 18:09:08.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/14/check/dsvm-tempest-kvm/8ca646b : FAILURE in 41m 01s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/14/check/dsvm-tempest-exe/06141d3 : FAILURE in 45m 23s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":14},{"id":"8190a3fd1bd20e4b203bb568f344d446ff2cb6fb","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-10 18:23:22.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/14/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c32eee4 : FAILURE in 1h 53m 43s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/14/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e58cd9a : SUCCESS in 1h 27m 48s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":14},{"id":"10fd515ddeb8b15c18d87e72afc17ed706cb5e0f","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-09-10 18:26:45.000000000","message":"Patch Set 14:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":14},{"id":"706a5a337d85d60d627fc5b4d81327a00ed4e168","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-09-10 18:46:16.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/14/check/dsvm-tempest-infinibox-fc/628307e : SUCCESS in 57m 23s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/14/check/dsvm-tempest-infinibox-iscsi/2389962 : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":14},{"id":"72d8ede7b26daeeb3139edfcb141cf58049141f0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-09-10 18:47:18.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"4d843f15f981966206413b107f9a4a5f64a73eaf","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-09-10 19:13:58.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/14/check/dsvm-tempest-zadara-driver/7cd82f8 : ABORTED in 46m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8d955724ce7a74adadf5a7c401ab8e0950f906b4","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-09-10 19:14:36.000000000","message":"Patch Set 14:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":14},{"id":"ebdec9be4a82e81505b43b4d36ba6ce9b66ba3dd","author":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"date":"2019-09-10 19:24:32.000000000","message":"Patch Set 13:\n\nBuild failed. To recheck use \u0027datacore: recheck\u0027. For information about DataCore CI see https://wiki.openstack.org/wiki/ThirdPartySystems/DataCore_CI\n\n- dsvm-tempest-datacore-iscsi-driver http://173.221.96.158/15/670715/13/check/dsvm-tempest-datacore-iscsi-driver/09cd58f : FAILURE in 1m 01s\n- dsvm-tempest-datacore-fc-driver http://173.221.96.158/15/670715/13/check/dsvm-tempest-datacore-fc-driver/7b83acf : FAILURE in 2h 11m 33s","accounts_in_message":[],"_revision_number":13},{"id":"6c1ce98cf0b78c03b86f1a2c98778d7bc71fef74","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-09-10 19:48:51.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"d5484b9dea4858469d35e60dabb3041e1e3c781b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-09-10 20:17:52.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/14/check/dsvm-tempest-zadara-driver/0683b9c : SUCCESS in 1h 03m 00s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"82e21b7c1e36a55b89de6f0a3bac6078879e917c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-09-10 20:36:38.000000000","message":"Patch Set 13:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"48561d33359a0b0f0fc31aaa8a46ea158233a154","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 21:14:11.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/14/check/NEC-ISCSI/afa8c18 : SUCCESS in 2h 01m 00s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/14/check/NEC-FC/738653f : SUCCESS in 2h 22m 55s","accounts_in_message":[],"_revision_number":14},{"id":"78a3ee58cc8fc0fa4db6ac7cd6ad96ddf257bd1e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 22:02:12.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/14/check/DellEMC_XtremIO_FC/b1b9dfc/DellEMC_XtremIO_FC/1411 : SUCCESS in 1h 29m 30s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/14/check/DellEMC_XtremIO_iSCSI/0486fcf/DellEMC_XtremIO_iSCSI/1325 : SUCCESS in 1h 38m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":14},{"id":"3f1d002cd934eeba325b29c209bb57e297363722","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-09-10 23:33:10.000000000","message":"Patch Set 14:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"e0af842f3e800ecc61aeb51bd1d55b501aa25c2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 00:32:53.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71c19be2337140259e23fad3be95e61c : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/442d62d0248d4df6b04c5efbb48ad904 : FAILURE in 9m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df34c668bede4bc78eacff4f40c0f041 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83496bbfa9c34adeac619286ae374e58 : SUCCESS in 11m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f65d2a8cc26248b6b023b19d2ecc6aa8 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab57b28154e547ae88e52ffb31175c90 : SUCCESS in 12m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bd0e0b5d141e430f918854d9f83a51c7 : SUCCESS in 1h 49m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dfe15d6576964963bd218d034f32f6c9 : SUCCESS in 56m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ee9852ebfc1e4ec38138687709bcbd3b : SUCCESS in 1h 00m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f346bde1896459c8a0188564cffca1c : SUCCESS in 1h 20m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32b6118d0a7c4080acbdaf51d0c5775b : SUCCESS in 18m 10s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae291b6e13034446974419c6b622b768 : SUCCESS in 1h 27m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aaa1971499ef4b449ddbf94610cb803e : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/825a66d5b2d64491a8999570a57e0b0c : SUCCESS in 5m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e329690d15a4fc48dadc09d0efc6236 : FAILURE in 35m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ce0b06c7db1e4d478a8c16066aedd03d : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b8133f36c0c490aaee47533b58e4d70 : SUCCESS in 1h 29m 16s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1d4755984b384f698d95a6414373991f : SUCCESS in 5m 17s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/adf947e5df944873ae80155532ecf7b6 : SUCCESS in 1h 12m 00s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1542c16e699d4ae883fe26c93d473fce : SUCCESS in 1h 39m 36s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/0bf85e3e4e4e4d81996a56f50514598c : SUCCESS in 1h 38m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b26b147e6a0420dbfa085e5640c844a : SUCCESS in 2h 24m 18s","accounts_in_message":[],"_revision_number":14},{"id":"fb343f4920550c3f8ddc7ac91afc38983a2c3669","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-11 02:39:51.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/14/check-not-vote/ext-cinder-zuul/d322daa : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":14},{"id":"11adeed676b0e4f6f75574726fbf319f80257d55","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-11 03:34:49.000000000","message":"Patch Set 14:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/14/check/tempest-dsvm-full-xenial/2e01386/ : SUCCESS in 2h 30m 02s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/14/check/devstack-plugin-ceph-dsvm-xenial/a99f1cb/ : SUCCESS in 2h 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e3ae7e13fda916e12f59bf7c9657479918233906","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-09-11 04:25:17.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/15/670715/14/check/cinder-storpool-tempest/1c4fef7/ : SUCCESS in 1h 27m 48s","accounts_in_message":[],"_revision_number":14},{"id":"c48f3184e41b1fd4c85078fd1d7364862cec601f","author":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"date":"2019-09-11 07:32:47.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"86986af4a3240d68f39ce5a4f732c94477860cdf","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-09-11 07:33:36.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"58af6c7eb5b513935b1676b1706315feacb02c57","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-09-11 07:34:33.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"5489d093e67974db2faec9ce8946fcfcf9e30238","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2019-09-11 07:42:37.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"b22e106582d0661b3cbde96c9c75116f0ca0093f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-09-11 07:53:28.000000000","message":"Patch Set 15:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/925/670715/15/check/infortrend-cinder-volume-iscsi/3ebefda51574b44349279d9421b22a593b14dd10 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/903/670715/15/check/infortrend-cinder-volume-fc/3ebefda51574b44349279d9421b22a593b14dd10 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"f782a6aab892a7a7e836082225ebb1e357e2ece0","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 08:13:38.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/15/check/dsvm-tempest-kvm/883fdfa : FAILURE in 34m 37s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/15/check/dsvm-tempest-exe/8ee3f6c : FAILURE in 39m 09s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"844ea297545b5a398e36d92ced393d089ff4056a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-09-11 08:37:51.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/15/check/dsvm-tempest-zadara-driver/e7495b8 : SUCCESS in 1h 04m 48s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1393d9f354e87e996ac6d65a82e6e2b1d738a09b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-11 08:44:09.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/15/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f45ab60 : SUCCESS in 1h 01m 43s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/15/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/c26133e : SUCCESS in 1h 11m 02s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":15},{"id":"efee996e2c2c20bca2f39a97af0fd8e4b0ca4dc6","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-09-11 08:45:01.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/15/iscsi/ : FAILURE in 1h 02m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/15/smb/ : FAILURE in 59m 52s","accounts_in_message":[],"_revision_number":15},{"id":"c277176d71662cde9ba37a66df3acc6ab843c133","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-09-11 09:12:00.000000000","message":"Patch Set 15:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/670715/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"15b1ad3f2ae84b35236cda421d88bf92b8d86522","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-09-11 09:57:21.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/15/670715/15/check/dsvm-tempest-infinibox-fc/ff098f1 : FAILURE in 17m 30s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/15/670715/15/check/dsvm-tempest-infinibox-iscsi/5e7c2a8 : SUCCESS in 1h 21m 17s","accounts_in_message":[],"_revision_number":15},{"id":"19b15650769040537e650bf4427ee28042299d62","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 10:00:18.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/15/670715/15/check/NEC-ISCSI/a3052ea : FAILURE in 7s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/15/670715/15/check/NEC-FC/a03e284 : SUCCESS in 2h 26m 50s","accounts_in_message":[],"_revision_number":15},{"id":"75e247850a081de54769b967da64fb25bd0f69d8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-11 10:36:31.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/15/check/tempest-dsvm-full-xenial/a843317/ : SUCCESS in 1h 57m 39s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/15/check/devstack-plugin-ceph-dsvm-xenial/3ad27f1/ : SUCCESS in 1h 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2aea47c81ed0679c7a9d95f753746425f0d5d8e5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-11 10:51:37.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/15/check-not-vote/ext-cinder-zuul/4ffe915 : SUCCESS in 49m 43s","accounts_in_message":[],"_revision_number":15},{"id":"223ae4d6e70f718fe0640718f96aca7bf62b1ab6","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-09-11 12:24:08.000000000","message":"Patch Set 15:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"307580c540a41d3ca7b0e503f8392ae07c688349","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 12:36:56.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a71832da32f04f7b9faee798fb5b64b6 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae2456338174cfeb267b6da682c6a37 : SUCCESS in 12m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eeaccf56978143c3a81267fadbd67c4d : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e916ca65d94417195a7ed39431f5ec9 : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a64fb27dbb14b15a16fb646d9600ab8 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40265f789e42457d995c94ca8fed0ef1 : SUCCESS in 13m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/59745fe0ff7847579d87adf34d57dc58 : SUCCESS in 1h 34m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0c039823ddea4fd2b21e702550a37bd8 : SUCCESS in 1h 01m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/995abbbe93974e0982f8f369360c8829 : SUCCESS in 1h 06m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8b35b284ff34a0592e6eaab724c0122 : SUCCESS in 1h 32m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbd4d9718cf345488badce214ccb84fe : SUCCESS in 14m 17s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bb230c652e1d472b95dca3d5dd2f99b0 : FAILURE in 1h 35m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9af3b1db890240d5894e6a15a11b4866 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0c6fcaaf7f7423da6b912a4b2eb9b74 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/235acfa0c1f84717bc695d98f6277532 : FAILURE in 39m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa4ce85e48a74ee4924bad623ef903ec : SUCCESS in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/736f757a2eeb46b7885d10ff8d788fa4 : SUCCESS in 1h 12m 54s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/74c3fc9e25ec411fbb12fd3d69807dd9 : SUCCESS in 6m 38s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/29b7b4eec2ed43e1ab282f33f208a6a7 : SUCCESS in 1h 23m 13s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ab285d37e36c4e8f86c5603f436fcf0a : SUCCESS in 1h 36m 25s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/6e41206d79864a1cbe85fe9f9dcb162f : SUCCESS in 1h 41m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/afd7487ead4146a1bd809f92d1191a6e : SUCCESS in 2h 18m 37s","accounts_in_message":[],"_revision_number":15},{"id":"246ab3ebe30136a009e7dd728d5ecbf2a9bc57f4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 14:20:24.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/15/check/DellEMC_XtremIO_FC/bc1f302/DellEMC_XtremIO_FC/1421 : SUCCESS in 1h 32m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/15/check/DellEMC_XtremIO_iSCSI/4029cf7/DellEMC_XtremIO_iSCSI/1335 : FAILURE in 1h 35m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"020f7fa80cdec07df2ae6eaee6394ddbb1d72dff","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2019-09-11 15:17:19.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"2e87bc584472a68dcf411a2cc136bffbf3d60717","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-09-11 15:19:04.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"62e5b33a501b9bac49f27e3804c8553abdd01618","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2019-09-11 15:58:45.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/15/670715/15/check/cinder-storpool-tempest/8669e09/ : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":15},{"id":"521442973af984568cbd3fce2ba34dbc533240c2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-09-11 17:10:16.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/670715/15/iscsi/ : FAILURE in 1h 30m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/670715/15/smb/ : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":15},{"id":"0b94f089adbb0cf4ca5998fb642ed0a50e5acf98","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 17:50:33.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/15/check/dsvm-tempest-kvm/eaebfe8 : FAILURE in 37m 42s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/15/670715/15/check/dsvm-tempest-exe/593e744 : FAILURE in 43m 06s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"7ff6334bd1f03e0e53c7599cab2928e69fbe0da2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-11 18:50:11.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/670715/15/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/2e5d410 : SUCCESS in 1h 03m 01s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/670715/15/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/621bfb6 : SUCCESS in 1h 13m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":15},{"id":"0dce7d33d883cff79cb1094f283aa701f5694c43","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-11 20:39:52.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/670715/15/check-not-vote/ext-cinder-zuul/49693e8 : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":15},{"id":"64c6c20e4865d19547ae694ddba12f1ed5b651a7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-12 00:28:52.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/15/check/tempest-dsvm-full-xenial/40be557/ : FAILURE in 38m 13s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/670715/15/check/devstack-plugin-ceph-dsvm-xenial/1b28f32/ : SUCCESS in 1h 45m 31s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a81ffa3cc2db1e89409c605d0b42135de7ab5201","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 00:33:54.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49ea8c0c0bc84a80b37b8da556a7ec06 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33142bbec7f456d97c908f3a40fe1b9 : SUCCESS in 8m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06cd260b6f034508882ebe51ff2f9dc3 : SUCCESS in 9m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35f1978228c54792a80e42a928494333 : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c20dbfebf664a6197a54b528ff683a4 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7c4208921e418d9583ee668807bc9a : SUCCESS in 14m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c5d9b2a58aae4c0981982eeeb9ec73d8 : SUCCESS in 1h 42m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7f2f75cb407a4beab13c91fa2075c34d : SUCCESS in 53m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b77f3878dfa4f26832a9caa5b1fe9fe : SUCCESS in 1h 02m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42725ea5e32f4269bac439ff5d9224fc : SUCCESS in 1h 35m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57e8ccf972424892bb70316d806ebd01 : SUCCESS in 18m 14s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7620c117ab614453bd159f8498e2c767 : SUCCESS in 1h 22m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9225c0d2dfd4e03b121b4809c092e0d : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/daf760b604444bb4ae9e976129bc9f57 : SUCCESS in 4m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a01e892df4824e579694be99e2fc0050 : FAILURE in 33m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4bffe732d0574fa19071bf773113e43e : SUCCESS in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f539cc8b656c4fa88cbd81e7a70bac2f : SUCCESS in 1h 16m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b4e589246d48484cb189422659ac7b68 : SUCCESS in 7m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/63e9462d558742cdb9a84ced099a025f : SUCCESS in 1h 19m 31s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c946cbf84114ea1b542f375021caa9c : SUCCESS in 1h 30m 13s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/133ff9797b2747bba8727abd3bdd2752 : SUCCESS in 1h 31m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/089088ad31d44986885ba7b105186ece : SUCCESS in 2h 25m 34s","accounts_in_message":[],"_revision_number":15},{"id":"b87b3f0a4a5934af981c57804505fb97122480b1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-09-12 01:41:00.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/15/670715/15/upstream-solidfire-check/cinder-solidfire-iSCSI/780f9e3/ : SUCCESS in 1h 03m 07s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"b038024c5c8e865dfdfe91c46142dedb7482a000","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-09-12 01:43:15.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/15/670715/15/check/synology-dsm-cinder-iscsi/d15d49e : FAILURE in 21m 16s","accounts_in_message":[],"_revision_number":15},{"id":"15123ab34e28c049e9061faa34ee74f963cb9dae","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 01:53:18.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/15/670715/15/check/DellEMC_PowerMAX_FC/fc06dd9/DellEMC_PowerMAX_FC/770 : FAILURE in 5h 45m 19s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/15/670715/15/check/DellEMC_PowerMAX_iSCSI/2cc2b7b/DellEMC_PowerMAX_iSCSI/715 : SUCCESS in 5h 05m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":15},{"id":"8efb5af48634c8a20c04131f57f304aae4665c0b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-09-12 02:58:37.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/670715/15/check/dsvm-tempest-zadara-driver/d632185 : SUCCESS in 1h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"86c5570cf79ae9cea4564906ce0b5ad4dfad8163","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-12 02:59:48.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/15/670715/15/upstream-check/cinder-cDOT-iSCSI/61705e6/ : SUCCESS in 51m 06s\n- cinder-cDOT-NFS http://13.56.159.105/logs/15/670715/15/upstream-check/cinder-cDOT-NFS/908cf70/ : SUCCESS in 52m 00s\n- cinder-cDOT-FCP http://13.56.159.105/logs/15/670715/15/upstream-check/cinder-cDOT-FCP/d2b3190/ : SUCCESS in 1h 08m 58s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"e24e697b9b3f72e223d8c493904b0caf4e134c70","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-09-12 03:42:55.000000000","message":"Patch Set 15:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/670715/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"1440fecde5f6d04a6f68548d4b5e40d8b465fbad","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-09-12 04:25:59.000000000","message":"Patch Set 15:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2477a4d : FAILURE in 1h 16m 53s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/15/670715/15/check/3par-fc-driver-master-client-pip-ssa02-dsvm/fdd2d7e : FAILURE in 1h 16m 49s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/15/670715/15/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/394f4f9 : FAILURE in 30m 10s","accounts_in_message":[],"_revision_number":15},{"id":"56a5958a9ca1ce6da5aae90da0628587a15538ff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-09-12 05:44:58.000000000","message":"Patch Set 15:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-15-670715-15/ns5_iscsi : FAILURE in 41m 33s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-15-670715-15/ns5_nfs : FAILURE in 40m 52s","accounts_in_message":[],"_revision_number":15},{"id":"4fd187635181e1882db37d3e19619ad356d2f0fa","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-09-12 14:55:17.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/15/670715/15/check/kaminario-dsvm-tempest-full-iscsi/0bef9c2 : SUCCESS in 1h 51m 25s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/15/670715/15/check/kaminario-dsvm-tempest-full-FC/ed926a1 : SUCCESS in 1h 37m 54s","accounts_in_message":[],"_revision_number":15},{"id":"573097975fd0688fac21477c7b8d7f7354f3bce6","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-09-12 17:56:05.000000000","message":"Patch Set 15:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/670715-15/Seagate-FC : SUCCESS in 38:17.91\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/670715-15/Seagate-iSCSI : SUCCESS in 41:48.66","accounts_in_message":[],"_revision_number":15},{"id":"ffc6e811e7a6c28ba89631f09e098732ebdb6d14","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 18:37:53.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/15/670715/15/check/DellEMC_VxFlexOS/3d18451/DellEMC_VxFlexOS/758 : SUCCESS in 1h 57m 24s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":15},{"id":"31949bc7ccc7e2ff2e6b1073c30a291e2d6caf24","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-09-12 20:21:06.000000000","message":"Patch Set 15:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/670715-15/HPMSA-FC : SUCCESS in 44:24.07\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/670715-15/HPMSA-iSCSI : SUCCESS in 50:16.17","accounts_in_message":[],"_revision_number":15},{"id":"1a5000252ad5210ca4a81e0732990fcaf74e28ac","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-09-12 21:18:09.000000000","message":"Patch Set 15:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/670715-15/Lenovo-FC : SUCCESS in 47:44.05\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/670715-15/Lenovo-iSCSI : SUCCESS in 49:56.03","accounts_in_message":[],"_revision_number":15},{"id":"074ff9caf8347c49c9ffb284c03c8b021fee0ac8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-09-12 21:52:29.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/670715/15/check/DellEMC_Unity_FC/4243c84/DellEMC_Unity_FC/1301 : SUCCESS in 1h 25m 36s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/670715/15/check/DellEMC_Unity_iSCSI/4fa256b/DellEMC_Unity_iSCSI/1245 : SUCCESS in 1h 42m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"ed98b84a1d65e870eb9ae1df5fa6618172f3b1d8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-09-12 22:16:31.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/670715/15/check/DellEMC_SC_FC/00b6526/DellEMC_SC_FC/643 : FAILURE in 1h 17m 24s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/670715/15/check/DellEMC_SC_iSCSI/e48a4c6/DellEMC_SC_iSCSI/617 : FAILURE in 2h 22m 54s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":15},{"id":"6f89d5c99a7fd8602b0c4d5851d9563731de028e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 23:38:52.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/670715/15/check/DellEMC_XtremIO_FC/2db520c/DellEMC_XtremIO_FC/1440 : SUCCESS in 1h 28m 17s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/670715/15/check/DellEMC_XtremIO_iSCSI/04218b9/DellEMC_XtremIO_iSCSI/1353 : SUCCESS in 1h 42m 16s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"091291c897ebfbacdbf8f0a660802e438aa52e8b","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-09-13 06:58:50.000000000","message":"Patch Set 15:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/15/2746 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"01994ffedefeff6e96295b21d67e712b4c15ea21","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-09-13 18:37:02.000000000","message":"Patch Set 15:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/942/670715/15/check/infortrend-cinder-volume-iscsi/3ebefda51574b44349279d9421b22a593b14dd10 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/920/670715/15/check/infortrend-cinder-volume-fc/3ebefda51574b44349279d9421b22a593b14dd10 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"11b06d4a221c22fe68105fd45a91519981d780c8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-19 14:27:24.000000000","message":"Patch Set 15: Code-Review+2\n\n(2 comments)\n\nOK, looks fine to me and CI is happy.","accounts_in_message":[],"_revision_number":15},{"id":"4eac2c9db246b37f40f3f6b3c8fb6cec7f54eb80","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-09-25 17:11:20.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\nLooks like comments have been addressed.  Should be good to go.","accounts_in_message":[],"_revision_number":15},{"id":"20bd84206ddb22f5ea20c2fdfa3881d18f1b3838","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 17:11:55.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"fce132f03d78d954ef8afdd3c9c62e9c7fe70b7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 19:38:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"8bec72c046b802e2390ef8dc69e4e3bb587ad7ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 19:38:02.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edf7aef9df624800841fa55455c6127a : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b277d2f031f4304b45732da2044bbb2 : SUCCESS in 7m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0420fbb300ef4f749ea465230afb4620 : SUCCESS in 11m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c73a839513224e6f8c741382b443a12b : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83702544add847ddbc1582f67d544dca : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9562e28f36a64ac9afea84407da21576 : SUCCESS in 14m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2f50ccf769ff481dae89179914edad5c : SUCCESS in 1h 25m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f3431add37da4e9c855002b19cc87087 : SUCCESS in 58m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/61f7c055546144c7b04ea4133e5f35b5 : SUCCESS in 55m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98a523222eac468baf4d2a677f99de96 : SUCCESS in 1h 19m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cdb8f00793f493283b5a42b02c3278a : SUCCESS in 16m 56s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d7d06f4cdb394f0fa39321ff87be3c50 : SUCCESS in 1h 10m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b46ebf98d13c4809a88d9e486e1b201f : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":15},{"id":"6404e99d54add87282990bad367e20e763ab8152","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 19:39:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a48f4ea937c4ca89649ca750eee1734 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/593abcb2e4c147328bd79cb25ec2ad82 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"3ebefda51574b44349279d9421b22a593b14dd10","revisions":{"1bdb60b7336f2416b14490deb68168e4db293204":{"kind":"REWORK","_number":1,"created":"2019-07-14 08:15:48.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/1"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:14:25.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nChange API authentication: Remove native \u0027username\\password\u0027\nlogin method and change it to \u0027access_key\u0027 auth. given\nas input. Fix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bdb60b7336f2416b14490deb68168e4db293204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bdb60b7336f2416b14490deb68168e4db293204"}]},"branch":"refs/heads/master"},"4ca599ecde51ae218c4c2b8fe786d2f229075a01":{"kind":"REWORK","_number":2,"created":"2019-07-15 07:11:53.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/2"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-15 07:10:08.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nChange API authentication: Remove native \u0027username\\password\u0027\nlogin method and change it to \u0027access_key\u0027 auth. given\nas input. Fix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ca599ecde51ae218c4c2b8fe786d2f229075a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ca599ecde51ae218c4c2b8fe786d2f229075a01"}]},"branch":"refs/heads/master"},"41fed678f495ddd031a025cf1cba369a5c10b19a":{"kind":"REWORK","_number":3,"created":"2019-07-23 14:43:46.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/3"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-23 14:42:03.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input. Fix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41fed678f495ddd031a025cf1cba369a5c10b19a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41fed678f495ddd031a025cf1cba369a5c10b19a"}]},"branch":"refs/heads/master"},"7cddbaa9c0c0b8f4253978368921738860f061f6":{"kind":"REWORK","_number":4,"created":"2019-07-24 08:32:18.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/4"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-24 08:31:54.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input. Fix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7cddbaa9c0c0b8f4253978368921738860f061f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7cddbaa9c0c0b8f4253978368921738860f061f6"}]},"branch":"refs/heads/master"},"87a7ac51dbcfc6103aaedf0e1acb6260bb9d0fee":{"kind":"REWORK","_number":5,"created":"2019-07-24 13:34:38.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/5"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-24 13:34:23.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input. Fix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87a7ac51dbcfc6103aaedf0e1acb6260bb9d0fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/87a7ac51dbcfc6103aaedf0e1acb6260bb9d0fee"}]},"branch":"refs/heads/master"},"d463d18b9ff9f40c3c6339e856365ac844088224":{"kind":"REWORK","_number":6,"created":"2019-07-25 05:54:43.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/6"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-25 05:54:16.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input. Fix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d463d18b9ff9f40c3c6339e856365ac844088224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d463d18b9ff9f40c3c6339e856365ac844088224"}]},"branch":"refs/heads/master"},"4fef05d78fa695b01ccf64f1afcef58ebbe1c9af":{"kind":"REWORK","_number":7,"created":"2019-07-25 13:39:29.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/7"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-25 13:39:14.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input. Fix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fef05d78fa695b01ccf64f1afcef58ebbe1c9af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fef05d78fa695b01ccf64f1afcef58ebbe1c9af"}]},"branch":"refs/heads/master"},"0ed9bf6faec9b0d33f98bff956275a8a251eec3b":{"kind":"REWORK","_number":8,"created":"2019-08-05 15:12:47.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/8"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-08-05 15:11:24.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input and move access_key to HTTP request header.\nFix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ed9bf6faec9b0d33f98bff956275a8a251eec3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ed9bf6faec9b0d33f98bff956275a8a251eec3b"}]},"branch":"refs/heads/master"},"ac3600dfacc2eb1bcd853841dd692060c88a68c5":{"kind":"REWORK","_number":9,"created":"2019-08-06 13:45:15.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/9"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-08-06 13:44:49.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input and move access_key to HTTP request header.\nFix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac3600dfacc2eb1bcd853841dd692060c88a68c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac3600dfacc2eb1bcd853841dd692060c88a68c5"}]},"branch":"refs/heads/master"},"4e24811f766e2ddb0425608e3df99ec227e80a87":{"kind":"REWORK","_number":10,"created":"2019-08-25 09:00:44.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/10"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-08-25 08:59:44.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input and move access_key to HTTP request header.\nFix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e24811f766e2ddb0425608e3df99ec227e80a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e24811f766e2ddb0425608e3df99ec227e80a87"}]},"branch":"refs/heads/master"},"fd7ef9a8348e9b345f329fa203e9fee3af5b2450":{"kind":"REWORK","_number":11,"created":"2019-08-28 09:36:17.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/11"}}},"commit":{"parents":[{"commit":"072dd9221b42849fcb33a4feb8ce1d3181d0a04d","subject":"Merge \"Docs: Add new driver review checklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/072dd9221b42849fcb33a4feb8ce1d3181d0a04d"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-08-28 09:35:40.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input and move access_key to HTTP request header.\nFix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd7ef9a8348e9b345f329fa203e9fee3af5b2450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd7ef9a8348e9b345f329fa203e9fee3af5b2450"}]},"branch":"refs/heads/master"},"25a920b4860d1c2cfc72d1f401c09fa1f91a0590":{"kind":"REWORK","_number":12,"created":"2019-09-05 06:36:11.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/12"}}},"commit":{"parents":[{"commit":"6e585da2ff1166c25ae7cf6f26ef1cd0c465dc37","subject":"Merge \"Move some code out of utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e585da2ff1166c25ae7cf6f26ef1cd0c465dc37"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-09-05 06:33:49.000000000","tz":180},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input and move access_key to HTTP request header.\nFix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25a920b4860d1c2cfc72d1f401c09fa1f91a0590"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25a920b4860d1c2cfc72d1f401c09fa1f91a0590"}]},"branch":"refs/heads/master"},"75b272548abb6829bfec0637396aaab4a1d4ca8d":{"kind":"REWORK","_number":13,"created":"2019-09-10 15:47:10.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/13"}}},"commit":{"parents":[{"commit":"6e585da2ff1166c25ae7cf6f26ef1cd0c465dc37","subject":"Merge \"Move some code out of utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e585da2ff1166c25ae7cf6f26ef1cd0c465dc37"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-09-10 15:47:06.000000000","tz":0},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input and move access_key to HTTP request header.\nFix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75b272548abb6829bfec0637396aaab4a1d4ca8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75b272548abb6829bfec0637396aaab4a1d4ca8d"}]},"branch":"refs/heads/master"},"39c1a0fd803a81435519393901767c6c9b624caf":{"kind":"REWORK","_number":14,"created":"2019-09-10 16:29:19.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/14"}}},"commit":{"parents":[{"commit":"6e585da2ff1166c25ae7cf6f26ef1cd0c465dc37","subject":"Merge \"Move some code out of utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e585da2ff1166c25ae7cf6f26ef1cd0c465dc37"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-09-10 16:27:26.000000000","tz":0},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input and move access_key to HTTP request header.\nFix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39c1a0fd803a81435519393901767c6c9b624caf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39c1a0fd803a81435519393901767c6c9b624caf"}]},"branch":"refs/heads/master"},"3ebefda51574b44349279d9421b22a593b14dd10":{"kind":"REWORK","_number":15,"created":"2019-09-11 07:32:47.000000000","uploader":{"_account_id":30629,"name":"Bhaa Shakur","email":"bhaa@zadara.com","username":"bshakur"},"ref":"refs/changes/15/670715/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/670715/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/670715/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/670715/15"}}},"commit":{"parents":[{"commit":"6e585da2ff1166c25ae7cf6f26ef1cd0c465dc37","subject":"Merge \"Move some code out of utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e585da2ff1166c25ae7cf6f26ef1cd0c465dc37"}]}],"author":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-07-14 08:08:28.000000000","tz":180},"committer":{"name":"Bhaa Shakur","email":"bhaa@zadara.com","date":"2019-09-11 07:31:05.000000000","tz":0},"subject":"Zadara VPSA: Move to API access key authentication","message":"Zadara VPSA: Move to API access key authentication\n\nAdd \u0027access_key\u0027 API authentication login method given\nas input and move access_key to HTTP request header.\nFix unittest accordingly.\n\nChange-Id: I11df21a7a213a0700d277d57ceee40f636955c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ebefda51574b44349279d9421b22a593b14dd10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ebefda51574b44349279d9421b22a593b14dd10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
