)]}'
{"id":"openstack%2Fcinder~671195","triplet_id":"openstack%2Fcinder~master~I3115ae296ae6b5702c7a8fa39249b8735542e17e","project":"openstack/cinder","branch":"master","topic":"rename-dothill-driver","hashtags":[],"change_id":"I3115ae296ae6b5702c7a8fa39249b8735542e17e","subject":"Create Seagate driver from dothill driver","status":"MERGED","created":"2019-07-17 06:44:43.000000000","updated":"2019-08-20 18:08:07.000000000","submitted":"2019-08-20 18:06:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671195-1566324361670-5b3f03e3","meta_rev_id":"b29721e8668a74c5dab295e22fab97ba5d84bbf2","_number":671195,"virtual_id_number":671195,"owner":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-19 16:38:04.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2019-08-17 20:24:11.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 19:07:56.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-17 21:18:27.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-17 20:02:46.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-20 18:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2019-08-17 20:35:05.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 20:01:11.000000000","_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},{"value":0,"date":"2019-08-19 17:41:58.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-08-19 14:28:55.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2019-08-17 00:24:02.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2019-08-19 17:33:49.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 18:02:40.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 15:18:03.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 18:28:11.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"date":"2019-08-19 18:27:22.000000000","_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"date":"2019-08-17 00:12:42.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-08-20 10:15:15.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 18:24:29.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2019-08-17 00:17:07.000000000","_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 17:35:06.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 18:07:55.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2019-08-19 18:10:37.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2019-08-16 23:50:47.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-17 02:49:54.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 17:54:41.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 18:06:20.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"date":"2019-08-17 00:18:45.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2019-08-17 01:55:58.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-19 18:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-20 16:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-20 16:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 18:58:44.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-08-13 17:47:26.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-08-14 03:26:03.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 03:48:09.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 19:01:13.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 21:49:16.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-08-16 21:52:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 22:49:44.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-08-16 23:50:47.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-17 00:12:42.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-08-17 00:17:07.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-17 00:18:45.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-08-17 00:24:02.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-08-17 01:55:58.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-17 02:49:54.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-17 20:02:46.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-17 20:24:11.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-17 20:35:05.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-17 21:18:27.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 14:28:55.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2019-08-19 15:18:03.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 16:38:04.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-08-19 17:33:49.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 17:35:06.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 17:41:58.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2019-08-19 17:54:41.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 18:02:40.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 18:06:20.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 18:07:55.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2019-08-19 18:10:37.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2019-08-19 18:24:29.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-08-19 18:27:22.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2019-08-19 18:28:11.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-08-19 18:42:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-19 19:07:56.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 10:15:15.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 16:13:56.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-08-20 18:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31b2bee093108fae699d63df283f2f6063d5ffba","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-07-17 06:44:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"303ae7ab6f90d51370508a6b76e40e9ff418d30a","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-17 07:04:22.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1764 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1715 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1915 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b8b4d195e5b32ac39864aef69b0f3342f5ce1688","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-17 07:25:27.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/95/671195/1/check-not-vote/ext-cinder-zuul/df1d25c : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5f07020798a12be3b020a0b14be8149b7a37346e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-17 07:40:03.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/1/iscsi/ : SUCCESS in 54m 46s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/1/smb/ : FAILURE in 51m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f730ba855b914ac06395c5a532d71e9a01121afa","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 08:13:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/1/check/dsvm-tempest-kvm/9d4337b : FAILURE in 39m 23s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/1/check/dsvm-tempest-exe/68150b9 : FAILURE in 40m 30s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"37897c3411c91b0de511bda9ffc9a19a1d48a664","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-17 08:14:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/1/check/tempest-dsvm-full-xenial/363667b/ : SUCCESS in 1h 13m 18s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/1/check/devstack-plugin-ceph-dsvm-xenial/daa61c2/ : SUCCESS in 1h 29m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9add9777d7c95529bc4b1cc8e05dc6fe4bf5f6db","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-17 08:34:20.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/1/check/fujitsu-eternus-dx-iscsi/01761ca : SUCCESS in 1h 47m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/1/check/fujitsu-eternus-dx-fc/53f3ad5 : SUCCESS in 1h 44m 31s","accounts_in_message":[],"_revision_number":1},{"id":"911717ca3e20df01960a1e06bbd5eaf83d82c62c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 09:08:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/95/671195/1/check/openstack-tox-lower-constraints/9c040c4/ : FAILURE in 18m 11s\n- openstack-tox-pep8 http://logs.openstack.org/95/671195/1/check/openstack-tox-pep8/2db70b9/ : FAILURE in 14m 56s\n- openstack-tox-py27 http://logs.openstack.org/95/671195/1/check/openstack-tox-py27/7e9650f/ : FAILURE in 13m 36s\n- openstack-tox-py36 http://logs.openstack.org/95/671195/1/check/openstack-tox-py36/01a23f4/ : FAILURE in 11m 35s\n- openstack-tox-py37 http://logs.openstack.org/95/671195/1/check/openstack-tox-py37/0bb25db/ : FAILURE in 9m 04s\n- openstack-tox-docs http://logs.openstack.org/95/671195/1/check/openstack-tox-docs/6f60b67/ : POST_FAILURE in 4m 22s\n- tempest-full http://logs.openstack.org/95/671195/1/check/tempest-full/5b5eb9c/ : SUCCESS in 1h 31m 56s\n- neutron-grenade http://logs.openstack.org/95/671195/1/check/neutron-grenade/7d2f026/ : SUCCESS in 1h 11m 19s\n- grenade-py3 http://logs.openstack.org/95/671195/1/check/grenade-py3/8412d75/ : SUCCESS in 55m 10s\n- tempest-full-py3 http://logs.openstack.org/95/671195/1/check/tempest-full-py3/5c480d0/ : SUCCESS in 1h 22m 06s\n- build-openstack-releasenotes http://logs.openstack.org/95/671195/1/check/build-openstack-releasenotes/f1210e4/html/ : SUCCESS in 16m 43s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/95/671195/1/check/cinder-tempest-dsvm-lvm-lio-barbican/305e48a/ : FAILURE in 1h 31m 12s\n- cinder-tox-bandit-baseline http://logs.openstack.org/95/671195/1/check/cinder-tox-bandit-baseline/ac1e4e8/ : FAILURE in 6m 31s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/95/671195/1/check/openstack-tox-functional-py36/dcc2cb5/ : SUCCESS in 4m 49s\n- cinder-rally-task http://logs.openstack.org/95/671195/1/check/cinder-rally-task/4c699b0/ : SUCCESS in 35m 41s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/95/671195/1/check/openstack-tox-pylint/e8f8ab7/ : FAILURE in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/95/671195/1/check/cinder-plugin-ceph-tempest/1824137/ : SUCCESS in 1h 17m 58s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/95/671195/1/check/openstack-tox-functional/c02b0f2/ : SUCCESS in 9m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/95/671195/1/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/91f7796/ : FAILURE in 10m 52s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/95/671195/1/check/legacy-tempest-dsvm-lvm-multibackend/e195141/ : FAILURE in 1h 33m 57s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/95/671195/1/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/c037660/ : SUCCESS in 1h 29m 40s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/95/671195/1/check/tempest-slow-py3/422d010/ : FAILURE in 2h 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"20105973cc92a5e4e67e17610cac2b2f2a2b5ea8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-17 09:08:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/1/check/DellEMC_Unity_FC/c6346aa/DellEMC_Unity_FC/692 : SUCCESS in 1h 05m 55s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/1/check/DellEMC_Unity_iSCSI/e98caa9/DellEMC_Unity_iSCSI/658 : SUCCESS in 1h 24m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"65276bf8b60d2d696ea78cce9d926cc867ec6a57","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 10:17:15.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/1/check/DellEMC_XtremIO_FC/85656ae/DellEMC_XtremIO_FC/863 : SUCCESS in 1h 31m 18s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/1/check/DellEMC_XtremIO_iSCSI/ec7765a/DellEMC_XtremIO_iSCSI/810 : SUCCESS in 1h 32m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"322c845468b7ce6dc4709e8fdaec104cb057f631","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-17 10:30:59.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3e353b15e664f494598825df008d944c291caead","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-07-17 11:46:51.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/95/671195/1/check/Cinder-AS13000/4c7c548 : SUCCESS in 2h 30m 51s\n- cinder-test-fc http://218.29.111.78:9000/95/671195/1/check/cinder-test-fc/1f3f0a5 : FAILURE in 8m 19s\n- cinder-test-iscsi http://218.29.111.78:9000/95/671195/1/check/cinder-test-iscsi/1b93b42 : FAILURE in 9m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0151ed26db9b684ca55c3ae75398182c8e4c343a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-17 13:16:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/671195/1/check/dsvm-tempest-cisco-zonemanager-job/8ae3c6b/ : SUCCESS in 1h 09m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cfb0e6b185b84be6c17c213c6bb9299ceab38f65","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-07-17 18:58:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nA couple of comments on the documentation.  We did agree in our meeting that this is ok to go forward with.","accounts_in_message":[],"_revision_number":1},{"id":"290b5ebfd6d6a86521f3cfdca5b1c3d957da7737","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-07-18 05:13:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce043dde344f03aee44a60dec62074c4ffaa9901","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-18 05:18:50.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1775 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1726 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1926 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"7c7e8914cf052572709eea823f3daacf509ddae2","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 06:00:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/2/check/dsvm-tempest-kvm/0681cdb : FAILURE in 34m 44s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/2/check/dsvm-tempest-exe/eab288e : FAILURE in 41m 20s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"f5d6af28930b763bab932f78cf0c1ed75984893a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-18 06:21:45.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/95/671195/2/check-cinder/Cinder-ISER-ISCSI/aaf5d64 : FAILURE in 1h 07m 36s\n- Cinder-ISER-LIO http://13.74.249.42/95/671195/2/check-cinder/Cinder-ISER-LIO/a4c9a7c : SUCCESS in 1h 00m 44s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"aa7fee67ee6ffdeb89246d326b894da7f146df5f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-18 06:39:07.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b8716b8d3903d5faad0e44f8558319352064de46","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-18 06:41:37.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/2/check/tempest-dsvm-full-xenial/9655331/ : SUCCESS in 1h 25m 57s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/2/check/devstack-plugin-ceph-dsvm-xenial/7ccd9e9/ : SUCCESS in 1h 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"14ded4950365dcf1fcc8d0931d459c9ea36b3123","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-18 06:51:19.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/2/check/fujitsu-eternus-dx-iscsi/5442569 : SUCCESS in 1h 37m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/2/check/fujitsu-eternus-dx-fc/cce8d88 : SUCCESS in 1h 36m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9ca4e8e3f291bfffd04ab21b8786bf0346e84527","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-18 07:07:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/671195/2/check/dsvm-tempest-cisco-zonemanager-job/27346e6/ : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a9e69a3886479777205b656543af1ca50562bdfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 07:47:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/95/671195/2/check/openstack-tox-lower-constraints/02ac42d/ : FAILURE in 14m 56s\n- openstack-tox-pep8 http://logs.openstack.org/95/671195/2/check/openstack-tox-pep8/c8b7d68/ : FAILURE in 26m 20s\n- openstack-tox-py27 http://logs.openstack.org/95/671195/2/check/openstack-tox-py27/cbb585a/ : FAILURE in 13m 53s\n- openstack-tox-py36 http://logs.openstack.org/95/671195/2/check/openstack-tox-py36/da7c169/ : FAILURE in 14m 18s\n- openstack-tox-py37 http://logs.openstack.org/95/671195/2/check/openstack-tox-py37/d5dcd0f/ : FAILURE in 9m 27s\n- openstack-tox-docs http://logs.openstack.org/95/671195/2/check/openstack-tox-docs/d4fee46/ : FAILURE in 10m 48s\n- tempest-full http://logs.openstack.org/95/671195/2/check/tempest-full/222919c/ : SUCCESS in 1h 30m 30s\n- neutron-grenade http://logs.openstack.org/95/671195/2/check/neutron-grenade/7096fb1/ : SUCCESS in 57m 52s\n- grenade-py3 http://logs.openstack.org/95/671195/2/check/grenade-py3/7dcf166/ : SUCCESS in 1h 04m 46s\n- tempest-full-py3 http://logs.openstack.org/95/671195/2/check/tempest-full-py3/e1ec83f/ : SUCCESS in 1h 36m 05s\n- build-openstack-releasenotes http://logs.openstack.org/95/671195/2/check/build-openstack-releasenotes/943dca0/html/ : SUCCESS in 16m 55s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/95/671195/2/check/cinder-tempest-dsvm-lvm-lio-barbican/18a387c/ : FAILURE in 1h 34m 39s\n- cinder-tox-bandit-baseline http://logs.openstack.org/95/671195/2/check/cinder-tox-bandit-baseline/fb3e2a5/ : FAILURE in 10m 42s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/95/671195/2/check/openstack-tox-functional-py36/67db4d9/ : SUCCESS in 10m 43s\n- cinder-rally-task http://logs.openstack.org/95/671195/2/check/cinder-rally-task/1ead850/ : SUCCESS in 50m 01s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/95/671195/2/check/openstack-tox-pylint/0ec4152/ : SUCCESS in 11m 01s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/95/671195/2/check/cinder-plugin-ceph-tempest/328db54/ : SUCCESS in 1h 15m 56s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/95/671195/2/check/openstack-tox-functional/85cb1fa/ : SUCCESS in 11m 15s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/95/671195/2/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/901d15f/ : SUCCESS in 1h 16m 21s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/95/671195/2/check/legacy-tempest-dsvm-lvm-multibackend/6594dcd/ : SUCCESS in 1h 30m 41s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/95/671195/2/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/3e69f8f/ : SUCCESS in 1h 33m 55s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/95/671195/2/check/tempest-slow-py3/3094f94/ : SUCCESS in 2h 21m 43s","accounts_in_message":[],"_revision_number":2},{"id":"28a2fa8f48daeecc146bf6c2b391ec309d69e417","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-18 08:26:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/2/check/DellEMC_Unity_FC/c248308/DellEMC_Unity_FC/708 : SUCCESS in 1h 07m 23s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/2/check/DellEMC_Unity_iSCSI/135982e/DellEMC_Unity_iSCSI/674 : SUCCESS in 1h 33m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1c51c0879f96427a9cc07fd91b85d2f7fdec3a9b","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-07-18 11:12:34.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/95/671195/2/check/Cinder-AS13000/ae2c96c : SUCCESS in 2h 32m 04s\n- cinder-test-fc http://218.29.111.78:9000/95/671195/2/check/cinder-test-fc/7136056 : FAILURE in 39m 13s\n- cinder-test-iscsi http://218.29.111.78:9000/95/671195/2/check/cinder-test-iscsi/35b9cff : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":2},{"id":"9bccce316f85bc9dc20650ad98dab02be5378e3d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-18 12:23:07.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/95/671195/2/check-not-vote/ext-cinder-zuul/1043ea2 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9d64c04f138d5c7f88d000de25f429c91743131d","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-18 14:14:38.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1726 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1778 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1928 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"17cf935ab0dfbf4e9d4837b470a2e2668a79377b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 15:49:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/2/check/DellEMC_XtremIO_FC/b9aa79e/DellEMC_XtremIO_FC/879 : SUCCESS in 1h 34m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/2/check/DellEMC_XtremIO_iSCSI/5263b9d/DellEMC_XtremIO_iSCSI/826 : FAILURE in 5h 05m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"9670d04200a80fa40f8cd8301ef564b93d3aee4f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 19:08:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/95/671195/2/check/DellEMC_PowerMAX_FC/1e1555a/DellEMC_PowerMAX_FC/438 : FAILURE in 4h 19m 35s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/95/671195/2/check/DellEMC_PowerMAX_iSCSI/cd7966d/DellEMC_PowerMAX_iSCSI/423 : FAILURE in 5h 17m 57s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"c3cfde2c94a2a93bf3b2ba95123ba757f7d9909d","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-07-21 20:02:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad87be366b99f5f3eb68c4a80737c8b09cbd9d48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-21 20:02:55.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"10c1741c1cff1540946acd554d3a6c3e6faaeec6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-21 20:03:23.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"1e9eedce0254bdc2a4eb55b156793e80b7d6600f","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-21 20:24:26.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1803 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1758 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1955 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"43894daecbb3bf7c03b0e1f95f8e4056ba52efb5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-21 20:43:37.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/3/check/dsvm-tempest-kvm/28ee9b0 : FAILURE in 35m 19s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/3/check/dsvm-tempest-exe/3179123 : FAILURE in 35m 25s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"316860add32767f73de0340b71614bbd23959e0a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-21 20:55:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/3/iscsi/ : SUCCESS in 52m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/3/smb/ : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":3},{"id":"a197ca0ec111661d1ff8affb2f29296e0a4dde6c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-21 20:57:46.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/3/check/dsvm-tempest-infinibox-fc/61bc36a : SUCCESS in 54m 36s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/3/check/dsvm-tempest-infinibox-iscsi/9c647bc : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"8addcc07fa95fb4334967f13f431bbdd20bef488","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-21 21:33:51.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/3/check/tempest-dsvm-full-xenial/4c1f6b3/ : SUCCESS in 1h 26m 36s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/3/check/devstack-plugin-ceph-dsvm-xenial/1f11aa4/ : SUCCESS in 1h 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c4274ec65208958ea4ca2981830787a80d3a044","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-21 21:36:17.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"8c50b4957264fb917b2e852a7b8bc19d4d12009f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-21 22:02:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/3/check/DellEMC_Unity_FC/79ec556/DellEMC_Unity_FC/748 : SUCCESS in 1h 01m 45s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/3/check/DellEMC_Unity_iSCSI/7d2bab9/DellEMC_Unity_iSCSI/713 : SUCCESS in 1h 30m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"f1d55d8268a26f774f26960a0539863a5b1b130f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-21 23:02:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/3/check/NEC-ISCSI/f067619 : SUCCESS in 2h 01m 08s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/3/check/NEC-FC/08abf61 : SUCCESS in 1h 59m 38s","accounts_in_message":[],"_revision_number":3},{"id":"29b4f82d7d5400c79dddc213e52905da31928fca","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-21 23:22:32.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/95/671195/3/check-not-vote/ext-cinder-zuul/edff70a : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":3},{"id":"0628ce8090b0febfded955bf88dc850eb67795f3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 00:04:08.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/95/671195/3/check/DellEMC_PowerMAX_FC/e9350d3/DellEMC_PowerMAX_FC/465 : FAILURE in 2h 31m 07s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/95/671195/3/check/DellEMC_PowerMAX_iSCSI/0595414/DellEMC_PowerMAX_iSCSI/446 : FAILURE in 2h 14m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"efd9b0e50141791ce620ee1b91719f77322181fa","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-07-22 02:38:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f1c5ac0023dde788da6b8b73376d50c773ac585c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 02:38:38.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"36d0ae7e5acc72f2fe232cbb47d53dca0bdf7cbb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-22 02:39:10.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"2ccde697cd0667a75048c2a7ad104fe2e4bb61b0","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-22 03:19:30.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1963 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1811 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1766 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"4e0eb3dd12f0ee6920ceae6e902b8d541e6a8ba4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-22 03:34:36.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/4/check/dsvm-tempest-infinibox-fc/0462fa5 : SUCCESS in 55m 16s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/4/check/dsvm-tempest-infinibox-iscsi/5957717 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"b91405d077eb70fc2f377a4a1f8d480a3660addf","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-22 04:01:33.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/4/iscsi/ : FAILURE in 1h 22m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/4/smb/ : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":4},{"id":"b5cc65ba8c77d82f99d0e1af89455fbfedd836f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-22 04:05:55.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/4/check/tempest-dsvm-full-xenial/95ed77b/ : SUCCESS in 1h 14m 29s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/4/check/devstack-plugin-ceph-dsvm-xenial/5ab7d8a/ : SUCCESS in 1h 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7cf62da816c591ce93a5da86ef8cf3ac284042c2","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 04:11:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/4/check/dsvm-tempest-kvm/228869a : FAILURE in 38m 54s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/4/check/dsvm-tempest-exe/8e03467 : FAILURE in 40m 26s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"6db320bb3fab2e3f8b1a43d04267cd482dded67c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-22 04:48:29.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"7c6c5bd50234cf4e5ab183423eba1955bc630bdc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-22 05:06:15.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/4/check/DellEMC_Unity_FC/088cbe0/DellEMC_Unity_FC/755 : FAILURE in 1h 20m 34s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/4/check/DellEMC_Unity_iSCSI/bb5a7ff/DellEMC_Unity_iSCSI/719 : FAILURE in 21m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"9ac9ff2f46d2bd972b3dce817962fd765d025b39","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-22 05:24:59.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/95/671195/4/check-not-vote/ext-cinder-zuul/24a8172 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":4},{"id":"fdc232447cfae575ea4819e492cbc2a1acc5a0f4","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-07-22 05:44:02.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9cabaec5be8a3d0f3578f6e9bd4564fb098f7d04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 05:44:16.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"a5559dfd6c0cfd4027dc3ddde86cabb9cd28789d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-22 05:44:50.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"a67e0bd15f33230ef8086b19cc7d3ee55a5327fc","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-07-22 05:59:14.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1767 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"430ababa637f6d15b71f7a70c10b23a5b1135111","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 06:30:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/5/check/dsvm-tempest-kvm/cc69cac : FAILURE in 33m 36s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/5/check/dsvm-tempest-exe/e12f941 : FAILURE in 40m 19s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"4d63f204f3f432854d5fab31bec82de34f518f5d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-22 06:35:25.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/5/iscsi/ : SUCCESS in 50m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/5/smb/ : FAILURE in 47m 42s","accounts_in_message":[],"_revision_number":5},{"id":"c155e755ca6890de871816693d345e35e6a5b68e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-22 06:40:10.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/5/check/dsvm-tempest-infinibox-fc/0de41d8 : SUCCESS in 55m 29s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/5/check/dsvm-tempest-infinibox-iscsi/3a6e0f5 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"2ca36d4430f43ee7911f6e5bf618c0f72e8f9c78","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-22 06:56:33.000000000","message":"Patch Set 5:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/95/671195/5/check-cinder/Cinder-ISER-ISCSI/dd9322f : FAILURE in 1h 11m 06s\n- Cinder-ISER-LIO http://13.74.249.42/95/671195/5/check-cinder/Cinder-ISER-LIO/ba12a42 : FAILURE in 1h 02m 14s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"1ea641d13b21510af9182188eb266d543f164be4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-22 07:06:04.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/5/check/DellEMC_Unity_FC/0bc53a6/DellEMC_Unity_FC/756 : FAILURE in 1h 14m 52s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/5/check/DellEMC_Unity_iSCSI/178c150/DellEMC_Unity_iSCSI/720 : SUCCESS in 1h 21m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"9854583f91c3a7207fe7690d1c150682e1d0392c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-22 07:12:51.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e7e2fd467c7c4b6d806b0460620d433462fa484d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 07:16:56.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/5/check/DellEMC_XtremIO_FC/5a53566/DellEMC_XtremIO_FC/894 : SUCCESS in 1h 29m 09s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/5/check/DellEMC_XtremIO_iSCSI/a28aa5e/DellEMC_XtremIO_iSCSI/833 : SUCCESS in 1h 32m 33s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"33790684a8e8aa9a07b16b5e8c6cbc3cbe5ebbf8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-22 07:16:57.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/5/check/tempest-dsvm-full-xenial/70066c4/ : SUCCESS in 1h 24m 49s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/5/check/devstack-plugin-ceph-dsvm-xenial/02db191/ : SUCCESS in 1h 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1ca142320f33480a2e0528cd56969e1a7f666c4d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-22 09:34:52.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/95/671195/5/check-not-vote/ext-cinder-zuul/035250d : SUCCESS in 43m 30s","accounts_in_message":[],"_revision_number":5},{"id":"4073ae8bc2daa69119e567a3c01e79d5a3860271","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-22 14:46:37.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/5/check/NEC-ISCSI/cb5494e : SUCCESS in 2h 15m 52s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/5/check/NEC-FC/f30218a : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":5},{"id":"b4d324e3e3c71859ba45dcbd7fdba106a977f3cf","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-07-23 16:08:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9f5b7e8b96b90f30cb68a94c7c49b79fdaefc28f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:08:48.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"37351a03813222f25b72ab5a7224dc41143f32a3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-23 16:09:20.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"8b5016c46847b4fe3f1f07a98321885854de0550","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-23 16:42:59.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1833 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1792 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1985 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"5926e75dffa1f720afc6ac182346155ebe0ad5ca","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-23 17:04:28.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/6/check/dsvm-tempest-infinibox-fc/829fa05 : SUCCESS in 54m 47s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/6/check/dsvm-tempest-infinibox-iscsi/ef9b5d5 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"9ce7f4baa88f45e3660b4b686714d3f40a6fa6ba","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-23 17:06:20.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/6/iscsi/ : SUCCESS in 57m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/6/smb/ : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":6},{"id":"4f16b9709350a737d16c97c968c72c54fe659a70","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-23 17:08:07.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/6/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/0251d28 : SUCCESS in 52m 06s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/6/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/58d22e3 : SUCCESS in 58m 27s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"f1fb0a53d5ddba05357d008d74542e17416c6247","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-23 17:38:56.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/6/check/dsvm-tempest-kvm/40c744f : FAILURE in 9m 15s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/6/check/dsvm-tempest-exe/b1d1653 : FAILURE in 39m 07s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"c174bfdc639a6a7ccb3215e96745f919fd8b7b8a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-23 17:57:18.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"77068d802ec89deebfb254f17246915ecb6fc5b4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-23 18:02:54.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/6/check/tempest-dsvm-full-xenial/212ea06/ : SUCCESS in 1h 29m 35s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/6/check/devstack-plugin-ceph-dsvm-xenial/f978c6d/ : SUCCESS in 1h 54m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"55491e01d00a229c2b02baf9efdf568dae088dd9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-23 18:46:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/95/671195/6/check/dsvm-tempest-zadara-driver/b1647db : SUCCESS in 1h 40m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f907cd8b7654bf03256f4b5a62a2582d346ad973","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-07-23 22:10:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/6/check/NEC-ISCSI/4b5c1b1 : SUCCESS in 1h 48m 47s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/6/check/NEC-FC/e2261cb : SUCCESS in 1h 58m 08s","accounts_in_message":[],"_revision_number":6},{"id":"eacfe5fe6c64a99c6d7934b185bcf34d0ed7d0c9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-24 00:08:39.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/6/check/DellEMC_Unity_FC/22807c3/DellEMC_Unity_FC/779 : SUCCESS in 1h 16m 17s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/6/check/DellEMC_Unity_iSCSI/8eda25d/DellEMC_Unity_iSCSI/741 : SUCCESS in 1h 33m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"450a5865fd08e65595262cca33c41d546ac91e4e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 01:25:53.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/6/check/DellEMC_XtremIO_FC/a9b7b63/DellEMC_XtremIO_FC/915 : SUCCESS in 1h 29m 19s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/6/check/DellEMC_XtremIO_iSCSI/80ec870/DellEMC_XtremIO_iSCSI/853 : SUCCESS in 1h 48m 08s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"9c9398b1a1bca920ca0690eded48ea81479299cc","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-24 02:30:02.000000000","message":"Patch Set 6:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/95/671195/6/check-cinder/Cinder-ISER-ISCSI/7ddcf29 : FAILURE in 3m 41s\n- Cinder-ISER-LIO http://13.74.249.42/95/671195/6/check-cinder/Cinder-ISER-LIO/4d24c85 : FAILURE in 3m 37s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"9e3764fba39eac599c77ac9d83e27c6fdd04725a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-24 04:24:24.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/95/671195/6/check-not-vote/ext-cinder-zuul/9af32c0 : FAILURE in 4m 02s","accounts_in_message":[],"_revision_number":6},{"id":"9caeb24b457bea6e4b7519604e9b2c0a83cd5388","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-06 21:50:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bbb37e53ee0c8f9f54212059eccd01fbbbbe00ed","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-06 22:14:22.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1928 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1942 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2093 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"44bd1bf5ab2e53117d6bce8e45a3ace18c26d375","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-06 22:45:18.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"e212c942dc8b602c07f55b650c0c9c484ac9e50f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-06 22:48:09.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/7/iscsi/ : SUCCESS in 56m 46s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/7/smb/ : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":7},{"id":"18e7db3f35fc8df6f9a754530403141e8162cf5c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-06 23:00:52.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/7/check/dsvm-tempest-infinibox-fc/fa436d1 : FAILURE in 1h 09m 48s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/7/check/dsvm-tempest-infinibox-iscsi/e8e69f2 : FAILURE in 1h 08m 24s","accounts_in_message":[],"_revision_number":7},{"id":"91e4a511f027499c0564243dc249005540d3fcc4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-06 23:28:31.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/7/check/fujitsu-eternus-dx-iscsi/c3598b9 : FAILURE in 1h 37m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/7/check/fujitsu-eternus-dx-fc/3080e31 : FAILURE in 2m 18s","accounts_in_message":[],"_revision_number":7},{"id":"de9e4653bb090f9196c293f43d1fa568d0881629","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-07 00:13:50.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/7/check/tempest-dsvm-full-xenial/c130ffe/ : SUCCESS in 1h 50m 12s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/7/check/devstack-plugin-ceph-dsvm-xenial/f0eb2cb/ : SUCCESS in 1h 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5ad492fc2ffdae42c34f84dc31dcd91204442f3e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-07 00:27:33.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/7/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/83712ee : FAILURE in 1h 49m 44s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/7/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/038dfba : FAILURE in 1h 49m 23s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"77be3bbd0bb882ce19a06e4b1169f7bf58f7e8b5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 00:31:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/7/check/dsvm-tempest-kvm/8bb6245 : FAILURE in 1m 15s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/7/check/dsvm-tempest-exe/f287f50 : FAILURE in 35m 59s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"c363544474d9e3a441e40f54d16c1e8d1035ccc6","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 00:58:59.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/7/check/NEC-ISCSI/c270ff8 : FAILURE in 2h 03m 43s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/7/check/NEC-FC/146b80c : FAILURE in 2h 02m 36s","accounts_in_message":[],"_revision_number":7},{"id":"e91f8b691e10396020859012a581b5c9e31465e3","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-07 01:25:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b2c0ea4caca65232f70ead595c2b960f53210864","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-07 01:25:56.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"11d0a436d2c8012f33c127d0b2709b0e5063c031","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-07 01:36:03.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1929 : ABORTED\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1943 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2094 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"41dde4835cf2b56f3088b257ea1abf992f1e179f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 02:09:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/8/check/dsvm-tempest-kvm/6a1b85e : FAILURE in 34m 13s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/8/check/dsvm-tempest-exe/9e258dc : FAILURE in 33m 44s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"6529b73fbfa864da5934ef6c3ef728f5948ba84c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-07 02:29:45.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/8/iscsi/ : FAILURE in 1h 03m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/8/smb/ : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":8},{"id":"e650c2dbf70fe64d28ab53bb4ed9ce0aa31aad13","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-07 02:35:53.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/8/check/dsvm-tempest-infinibox-fc/31fa306 : FAILURE in 1h 03m 17s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/8/check/dsvm-tempest-infinibox-iscsi/34582df : FAILURE in 1h 10m 03s","accounts_in_message":[],"_revision_number":8},{"id":"879dfd91c31c32ad7df7ce35468c9252dead3771","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-07 02:37:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"73d5b56c41b516004296032f519f29732417241d","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-07 03:01:46.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2097 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1946 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1932 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"a178511aeb72ee8d94c56b925496442961599ecf","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 03:24:41.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/9/check/dsvm-tempest-kvm/989d8ac : FAILURE in 37m 27s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/9/check/dsvm-tempest-exe/fa0d06a : FAILURE in 38m 31s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"b5ebc0f9c4aa27a796c7ada5e57788c210df610f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-07 03:30:23.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"728f0841266195a95e3c1edfc14644741a32a8fe","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-07 03:37:12.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/9/iscsi/ : SUCCESS in 55m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/9/smb/ : FAILURE in 59m 32s","accounts_in_message":[],"_revision_number":9},{"id":"c4c2759cf849c0136781b02977b2ab159c4165e2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-07 03:43:16.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/9/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/2274303 : FAILURE in 52m 34s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/9/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/19251ef : FAILURE in 1h 05m 44s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"1e8ae25c8719e4a3394670bad560811170deee5d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-07 03:47:27.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/9/check/dsvm-tempest-infinibox-fc/eaabd32 : FAILURE in 54m 28s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/9/check/dsvm-tempest-infinibox-iscsi/f53d698 : FAILURE in 1h 08m 05s","accounts_in_message":[],"_revision_number":9},{"id":"708481a5537b13b84d0efed61ca924cba4136b53","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-07 03:57:34.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/9/check/tempest-dsvm-full-xenial/0f3453b/ : SUCCESS in 1h 20m 05s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/9/check/devstack-plugin-ceph-dsvm-xenial/7562fbd/ : SUCCESS in 1h 16m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"df5ff2fc4fe02214d0ca40c11ebd10256b1e6aa0","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-07 04:18:02.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/9/check/fujitsu-eternus-dx-iscsi/7839860 : FAILURE in 1h 39m 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/9/check/fujitsu-eternus-dx-fc/af164bf : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":9},{"id":"214fc0aa948fa6bb3bd7558239c02a684e35574c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 04:41:57.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/9/check/NEC-ISCSI/a6687cb : FAILURE in 2h 04m 22s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/9/check/NEC-FC/0c34d8f : FAILURE in 2h 03m 37s","accounts_in_message":[],"_revision_number":9},{"id":"a09dcb0af43d8cde30b13c784954bc9851e9a75c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 04:59:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/95/671195/9/check/openstack-tox-lower-constraints/da2a5f9/ : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://logs.opendev.org/95/671195/9/check/openstack-tox-pep8/5e14156/ : SUCCESS in 10m 01s\n- openstack-tox-py27 https://logs.opendev.org/95/671195/9/check/openstack-tox-py27/f0f0e93/ : SUCCESS in 11m 24s\n- openstack-tox-py36 https://logs.opendev.org/95/671195/9/check/openstack-tox-py36/a3ace7f/ : SUCCESS in 11m 03s\n- openstack-tox-py37 https://logs.opendev.org/95/671195/9/check/openstack-tox-py37/a31b859/ : SUCCESS in 10m 40s\n- openstack-tox-docs https://logs.opendev.org/95/671195/9/check/openstack-tox-docs/566b028/html/ : SUCCESS in 8m 41s\n- tempest-full https://logs.opendev.org/95/671195/9/check/tempest-full/9c528f5/ : SUCCESS in 1h 55m 04s\n- neutron-grenade https://logs.opendev.org/95/671195/9/check/neutron-grenade/9f2e5ab/ : SUCCESS in 56m 04s\n- grenade-py3 https://logs.opendev.org/95/671195/9/check/grenade-py3/5b435a6/ : SUCCESS in 1h 07m 51s\n- tempest-integrated-storage https://logs.opendev.org/95/671195/9/check/tempest-integrated-storage/b660530/ : SUCCESS in 1h 18m 15s\n- build-openstack-releasenotes https://logs.opendev.org/95/671195/9/check/build-openstack-releasenotes/9fb779d/html/ : SUCCESS in 14m 52s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/95/671195/9/check/cinder-tempest-dsvm-lvm-lio-barbican/36c34e0/ : FAILURE in 1h 26m 25s\n- cinder-tox-bandit-baseline https://logs.opendev.org/95/671195/9/check/cinder-tox-bandit-baseline/9b91891/ : FAILURE in 7m 49s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/95/671195/9/check/openstack-tox-functional-py36/0031535/ : SUCCESS in 6m 52s\n- cinder-rally-task https://logs.opendev.org/95/671195/9/check/cinder-rally-task/a213f94/ : SUCCESS in 48m 29s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/95/671195/9/check/openstack-tox-pylint/cc01243/ : SUCCESS in 7m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/95/671195/9/check/cinder-plugin-ceph-tempest/237d763/ : SUCCESS in 1h 17m 23s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/95/671195/9/check/openstack-tox-functional/1c81308/ : SUCCESS in 6m 01s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/95/671195/9/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/dd42b55/ : SUCCESS in 1h 02m 54s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/95/671195/9/check/legacy-tempest-dsvm-lvm-multibackend/bd36492/ : SUCCESS in 1h 20m 03s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/95/671195/9/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/99c2f19/ : SUCCESS in 1h 17m 28s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/95/671195/9/check/tempest-slow-py3/d3336bc/ : SUCCESS in 2h 16m 15s","accounts_in_message":[],"_revision_number":9},{"id":"0c8e8de67e1c54fc655a555935d5d8df770169b9","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-07 05:53:32.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2e2b64182ce3eba852ca21f74b4a25499a3c5458","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-07 06:18:26.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2099 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1951 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1935 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"88db84e048a1aebfc3d9382d377b3ea3dce708cf","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 06:28:21.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/10/check/dsvm-tempest-kvm/7763f31 : FAILURE in 34m 16s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/10/check/dsvm-tempest-exe/d513b37 : FAILURE in 1m 05s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"122bfa17e48b138c59e381bdac8a381907c5b331","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-07 06:47:59.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"b4a9f8bb714fa99fe54ac225a08f75923cfbba05","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-07 06:48:44.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/10/iscsi/ : SUCCESS in 53m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/10/smb/ : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":10},{"id":"fee8223e7f477d34598285a63f7de117806bb779","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-07 07:01:14.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/10/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/4f0548c : FAILURE in 1h 01m 22s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/10/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/59314d2 : SUCCESS in 1h 07m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"8824237e86cadfbc47890e5c5a6d90b5ad0dde80","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-07 07:08:24.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/95/671195/10/check-cinder/Cinder-ISER-ISCSI/a0c433f : SUCCESS in 1h 04m 17s\n- Cinder-ISER-LIO http://13.74.249.42/95/671195/10/check-cinder/Cinder-ISER-LIO/5f9fcf2 : SUCCESS in 1h 13m 51s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"5af7bfcf6ca2e0bfbd06e70fed1a837aefd31c4b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-07 07:08:32.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/95/671195/10/check-not-vote/ext-cinder-zuul/ff1936d : FAILURE in 2m 23s","accounts_in_message":[],"_revision_number":10},{"id":"878119f97f7af76783ea5518bd6142d87568b1a0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-07 07:10:23.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/10/check/dsvm-tempest-infinibox-fc/c3cd595 : SUCCESS in 56m 28s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/10/check/dsvm-tempest-infinibox-iscsi/dacc9be : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":10},{"id":"d89a248d2768b600748f51c2e4f98481859f1170","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-07 07:27:04.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/10/check/tempest-dsvm-full-xenial/29c85f5/ : SUCCESS in 1h 25m 49s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/10/check/devstack-plugin-ceph-dsvm-xenial/be69484/ : SUCCESS in 1h 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dbf58885f1f63aaab350825289f8f059d203b853","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 08:10:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/95/671195/10/check/openstack-tox-lower-constraints/7e7bcb6/ : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://logs.opendev.org/95/671195/10/check/openstack-tox-pep8/d4c49cc/ : SUCCESS in 8m 47s\n- openstack-tox-py27 https://logs.opendev.org/95/671195/10/check/openstack-tox-py27/a4af57d/ : SUCCESS in 11m 05s\n- openstack-tox-py36 https://logs.opendev.org/95/671195/10/check/openstack-tox-py36/2c68770/ : SUCCESS in 10m 50s\n- openstack-tox-py37 https://logs.opendev.org/95/671195/10/check/openstack-tox-py37/ba06015/ : SUCCESS in 10m 53s\n- openstack-tox-docs https://logs.opendev.org/95/671195/10/check/openstack-tox-docs/76a3969/html/ : SUCCESS in 10m 16s\n- tempest-full https://logs.opendev.org/95/671195/10/check/tempest-full/eb44623/ : SUCCESS in 1h 31m 59s\n- neutron-grenade https://logs.opendev.org/95/671195/10/check/neutron-grenade/7546264/ : SUCCESS in 1h 07m 06s\n- grenade-py3 https://logs.opendev.org/95/671195/10/check/grenade-py3/c788d2d/ : SUCCESS in 1h 14m 44s\n- tempest-integrated-storage https://logs.opendev.org/95/671195/10/check/tempest-integrated-storage/8204767/ : SUCCESS in 1h 19m 27s\n- build-openstack-releasenotes https://logs.opendev.org/95/671195/10/check/build-openstack-releasenotes/6eff80b/html/ : SUCCESS in 15m 05s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/95/671195/10/check/cinder-tempest-dsvm-lvm-lio-barbican/6fefc8a/ : SUCCESS in 1h 32m 20s\n- cinder-tox-bandit-baseline https://logs.opendev.org/95/671195/10/check/cinder-tox-bandit-baseline/263f237/ : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/95/671195/10/check/openstack-tox-functional-py36/cfe7b8f/ : SUCCESS in 4m 30s\n- cinder-rally-task https://logs.opendev.org/95/671195/10/check/cinder-rally-task/64c71b2/ : SUCCESS in 41m 54s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/95/671195/10/check/openstack-tox-pylint/2dd36e0/ : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/95/671195/10/check/cinder-plugin-ceph-tempest/f16bee9/ : SUCCESS in 1h 11m 18s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/95/671195/10/check/openstack-tox-functional/51ccaf2/ : SUCCESS in 4m 48s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/95/671195/10/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/62c86e8/ : SUCCESS in 1h 08m 07s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/95/671195/10/check/legacy-tempest-dsvm-lvm-multibackend/1d221e1/ : SUCCESS in 1h 24m 47s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/95/671195/10/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2fca7a1/ : SUCCESS in 1h 28m 36s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/95/671195/10/check/tempest-slow-py3/c0c3b76/ : SUCCESS in 2h 14m 57s","accounts_in_message":[],"_revision_number":10},{"id":"a5c0f7a99903a0eb8405ff49e2ee7df71460bf6e","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-07 08:41:43.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/95/671195/10/check/synology-dsm-cinder-iscsi/a21832b : FAILURE in 30m 58s","accounts_in_message":[],"_revision_number":10},{"id":"14d9c32b5450f69530ae7b76080cfab4aff06018","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 08:52:00.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/10/check/NEC-ISCSI/2e45803 : SUCCESS in 2h 06m 26s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/10/check/NEC-FC/fdb647f : SUCCESS in 2h 05m 49s","accounts_in_message":[],"_revision_number":10},{"id":"0f3888e49db6caeeddbb95b123e062f4cd9e28e5","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-07 09:07:50.000000000","message":"Patch Set 10:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/95/2466 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"8f827731392d9ae41a8799e82e138d530b79273a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-07 09:09:37.000000000","message":"Patch Set 10:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-95-671195-10 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":10},{"id":"9c03bed37df67a5938fa2f801ad5816034070e35","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-07 09:17:55.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/95/671195/10/upstream-solidfire-check/cinder-solidfire-iSCSI/ae8283b/ : SUCCESS in 1h 06m 25s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"587bbac723003109c3932ca30127faa611409ee0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-07 09:24:53.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.161.150/logs/95/671195/10/upstream-check/cinder-cDOT-iSCSI/ddfdc2b/ : SUCCESS in 54m 47s\n- cinder-cDOT-NFS http://13.56.161.150/logs/95/671195/10/upstream-check/cinder-cDOT-NFS/247397e/ : FAILURE in 2m 42s\n- cinder-cDOT-FCP http://13.56.161.150/logs/95/671195/10/upstream-check/cinder-cDOT-FCP/d35bd1a/ : SUCCESS in 1h 13m 54s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"7696b94e5abdf0f81f82d062aa7160cd8ba6ba97","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-07 09:37:05.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/95/671195/10/check/dsvm-tempest-zadara-driver/1286c90 : SUCCESS in 1h 41m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dfa0afc7328f698dea7cfc799464426d7685a8ce","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-07 09:47:32.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/95/671195/10/check/kaminario-dsvm-tempest-full-iscsi/e387d8d : SUCCESS in 1h 36m 54s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/95/671195/10/check/kaminario-dsvm-tempest-full-FC/919cd51 : FAILURE in 2m 14s","accounts_in_message":[],"_revision_number":10},{"id":"b918b10d5b4542d765c3177bbd726792f2a44dc1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 09:48:33.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/95/671195/10/check/DellEMC_VNX_iSCSI/bf44ef8/DellEMC_VNX_iSCSI/537 : SUCCESS in 1h 38m 04s\n- DellEMC_VNX_FC http://publiclogs.emc.com/95/671195/10/check/DellEMC_VNX_FC/48ca89b/DellEMC_VNX_FC/547 : SUCCESS in 1h 00m 29s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"78de97c642df969aa1a7b82cd9edfe45560df195","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-07 10:12:11.000000000","message":"Patch Set 10:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8b89409 : FAILURE in 1h 59m 25s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/10/check/3par-fc-driver-master-client-pip-ssa02-dsvm/e4ac629 : SUCCESS in 1h 59m 11s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/95/671195/10/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/97ff065 : FAILURE in 1h 29m 09s","accounts_in_message":[],"_revision_number":10},{"id":"ffd1ad054d894a67ada3a207f8a5629f4298b01c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-07 11:24:54.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/95/671195/10/check/DellEMC_SC_FC/12f9005/DellEMC_SC_FC/402 : SUCCESS in 1h 14m 43s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/95/671195/10/check/DellEMC_SC_iSCSI/a8392fe/DellEMC_SC_iSCSI/384 : SUCCESS in 1h 35m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"d6edafde8238749113ee883932b5a26f4afc5368","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-07 12:29:54.000000000","message":"Patch Set 10:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/671195-10/HPMSA-FC : FAILURE in 59:50.52\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/671195-10/HPMSA-iSCSI : SUCCESS in 47:23.84","accounts_in_message":[],"_revision_number":10},{"id":"1657e16d0762f57a4e8d9cae15b2f96be7a3fe79","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 12:46:01.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/10/check/DellEMC_XtremIO_FC/8cabdab/DellEMC_XtremIO_FC/1069 : SUCCESS in 1h 30m 21s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/10/check/DellEMC_XtremIO_iSCSI/7a0d7c8/DellEMC_XtremIO_iSCSI/991 : SUCCESS in 1h 26m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"2708b5bdd0c44998b70f09455e34e7173892927d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-07 14:41:37.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"c1759d5abc00a114aafcad839782669e98fb100b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-07 14:54:37.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"e03ffbd2f6c8b81657c88c6a98d6c4320747e85d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-07 16:17:39.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"2255d6fb0615299059bd1157cc37ffe250bfcea0","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-07 16:22:58.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"83dabe65664ece243881873353f48d740c3f6698","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 23:42:48.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/95/671195/10/check/DellEMC_VxFlexOS/0b66543/DellEMC_VxFlexOS/523 : ABORTED in 4h 00m 10s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":10},{"id":"6eb1ef06d79eba55edc7e162aff1b60473ef85d9","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-08 00:15:34.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5f28de13a5d288aed628d2cf1133e27f0984d005","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-08 00:21:16.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"1aa2dba4e1c893033990c4e78594c69d29a5a3a7","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-08 00:40:27.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2105 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1957 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1941 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"8a4c8b3fbb5bec4139b030313f129003b77f03de","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 00:49:59.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/11/check/dsvm-tempest-kvm/ee3dc1a : FAILURE in 33m 49s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/11/check/dsvm-tempest-exe/43c9935 : FAILURE in 40s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"0ce5bcb8a50db28427f5b8fcf53cb42e7a251a85","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-08-08 01:00:33.000000000","message":"Patch Set 10:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/671195-10/Lenovo-FC : FAILURE in 56:26.70\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/671195-10/Lenovo-iSCSI : FAILURE in 2:35:43","accounts_in_message":[],"_revision_number":10},{"id":"12987991ba58b00824559247f1e6be4cadf44817","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-08 01:25:59.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/11/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/dbf6635 : SUCCESS in 57m 59s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/11/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/f8161c3 : SUCCESS in 1h 10m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":11},{"id":"ccbe2805da6842f7d064ff2b40138a641a5ab3ef","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-08 01:31:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/11/check/dsvm-tempest-infinibox-fc/3722c33 : SUCCESS in 56m 31s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/11/check/dsvm-tempest-infinibox-iscsi/73c4b87 : SUCCESS in 1h 15m 11s","accounts_in_message":[],"_revision_number":11},{"id":"d38814d2a6f5f040126460353a404443288220ca","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-08 01:40:32.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/11/check/DellEMC_Unity_FC/1683b25/DellEMC_Unity_FC/946 : FAILURE in 1h 11m 28s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/11/check/DellEMC_Unity_iSCSI/ba0da75/DellEMC_Unity_iSCSI/904 : SUCCESS in 1h 24m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"965cef9ac45f4a0e7ac97e112141644536cafee0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 01:42:16.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/11/check/DellEMC_XtremIO_FC/3c0f0ac/DellEMC_XtremIO_FC/1075 : SUCCESS in 1h 26m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/11/check/DellEMC_XtremIO_iSCSI/6c82396/DellEMC_XtremIO_iSCSI/997 : SUCCESS in 1h 26m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":11},{"id":"6d3e95896c7956f8ac8e1e7e449769a178136cd1","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-08 01:57:40.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/95/671195/11/check/dsvm-tempest-zadara-driver/1366907 : SUCCESS in 1h 41m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a0a8ac0d2cd4573f947535a1c907bc502f41b849","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-08 02:10:05.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/11/check/fujitsu-eternus-dx-iscsi/dda15ab : SUCCESS in 1h 52m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/11/check/fujitsu-eternus-dx-fc/a3eca07 : FAILURE in 2m 40s","accounts_in_message":[],"_revision_number":11},{"id":"67e880b48b7e853efff93c2d9f392b3aecb589d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 05:00:12.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/95/671195/11/check/openstack-tox-lower-constraints/38a3143/ : SUCCESS in 16m 13s\n- openstack-tox-pep8 https://logs.opendev.org/95/671195/11/check/openstack-tox-pep8/c441fa6/ : FAILURE in 9m 13s\n- openstack-tox-py27 https://logs.opendev.org/95/671195/11/check/openstack-tox-py27/dee40ac/ : SUCCESS in 8m 57s\n- openstack-tox-py36 https://logs.opendev.org/95/671195/11/check/openstack-tox-py36/aa9caa3/ : SUCCESS in 8m 48s\n- openstack-tox-py37 https://logs.opendev.org/95/671195/11/check/openstack-tox-py37/fcd2e11/ : SUCCESS in 7m 47s\n- openstack-tox-docs https://logs.opendev.org/95/671195/11/check/openstack-tox-docs/1981fd0/html/ : SUCCESS in 8m 49s\n- tempest-full https://logs.opendev.org/95/671195/11/check/tempest-full/90f9dc4/ : SUCCESS in 1h 38m 59s\n- neutron-grenade https://logs.opendev.org/95/671195/11/check/neutron-grenade/11669c3/ : SUCCESS in 1h 00m 41s\n- grenade-py3 https://logs.opendev.org/95/671195/11/check/grenade-py3/6020e79/ : SUCCESS in 1h 06m 32s\n- tempest-integrated-storage https://logs.opendev.org/95/671195/11/check/tempest-integrated-storage/dda03de/ : SUCCESS in 1h 38m 37s\n- build-openstack-releasenotes https://logs.opendev.org/95/671195/11/check/build-openstack-releasenotes/4100425/html/ : SUCCESS in 14m 27s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/95/671195/11/check/cinder-tempest-dsvm-lvm-lio-barbican/538818e/ : FAILURE in 2h 53m 01s\n- cinder-tox-bandit-baseline https://logs.opendev.org/95/671195/11/check/cinder-tox-bandit-baseline/a76e087/ : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/95/671195/11/check/openstack-tox-functional-py36/9b74030/ : SUCCESS in 6m 12s\n- cinder-rally-task https://logs.opendev.org/95/671195/11/check/cinder-rally-task/b51fef2/ : SUCCESS in 43m 46s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/95/671195/11/check/openstack-tox-pylint/40ded63/ : SUCCESS in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/95/671195/11/check/cinder-plugin-ceph-tempest/4587cc0/ : SUCCESS in 1h 15m 33s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/95/671195/11/check/openstack-tox-functional/7ad02d7/ : SUCCESS in 6m 09s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/95/671195/11/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/d4d96a4/ : SUCCESS in 1h 08m 45s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/95/671195/11/check/legacy-tempest-dsvm-lvm-multibackend/c90057f/ : SUCCESS in 1h 18m 30s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/95/671195/11/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/9c31139/ : SUCCESS in 1h 53m 18s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/95/671195/11/check/tempest-slow-py3/dbeef8f/ : SUCCESS in 2h 16m 03s","accounts_in_message":[],"_revision_number":11},{"id":"a15eab2653428d9788d7c6d21a48edd22a904cb7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 05:59:21.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/11/check/NEC-ISCSI/4160d37 : SUCCESS in 2h 11m 25s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/11/check/NEC-FC/e2e4c29 : SUCCESS in 2h 06m 34s","accounts_in_message":[],"_revision_number":11},{"id":"2d9c1ad6899f161d735d1955280a60352f7153bf","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-08 12:52:29.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/11/iscsi/ : FAILURE in 13m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/11/smb/ : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":11},{"id":"c18deb72ef9fae54a3c7f5be51cfc4d96b5356c7","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-08 13:37:26.000000000","message":"Patch Set 11:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"1c505e9448173a1f35d8c4db254ac0bf130fd26d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 16:34:14.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/95/671195/11/check/DellEMC_PowerMAX_FC/99a5c3a/DellEMC_PowerMAX_FC/591 : FAILURE in 3h 23m 25s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/95/671195/11/check/DellEMC_PowerMAX_iSCSI/953a491/DellEMC_PowerMAX_iSCSI/562 : SUCCESS in 4h 47m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":11},{"id":"40555b6651d13e9a3b09049ab89f15aeb272b977","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-08 19:27:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7da5885634e42a49ca2aaf469ee7b5468190adc8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-08 19:29:40.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/12/iscsi/ : FAILURE in 1m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/12/smb/ : FAILURE in 1m 32s","accounts_in_message":[],"_revision_number":12},{"id":"fdf214bc0697282e4bd42e29fa6604541f9c08c9","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-08 19:52:06.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1922 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1931 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2082 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"bde2c90ba9c2ece93d991ef55f9b6d5472cccfba","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-08 20:22:03.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"e9fd412ac26d9d37b3e4e01f393d2d88d5153250","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-08 20:29:46.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/12/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/9d91cd5 : FAILURE in 54m 24s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/12/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/856a4e6 : SUCCESS in 1h 01m 49s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":12},{"id":"07b646414bc3caf15de34d13b07db5346a62e6ef","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-08 20:35:22.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/12/check/dsvm-tempest-infinibox-fc/3ba9a0c : SUCCESS in 48m 38s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/12/check/dsvm-tempest-infinibox-iscsi/7bb9104 : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":12},{"id":"f4d83142a55a309c5c30a88c77cecb314ab7382e","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-08 21:02:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"42932c47658af3325837666b20b2ecee6ef68a38","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-08 21:28:04.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1923 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1932 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2083 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"fbb38d4b173f23923f74c589bfd11bc4dfc2c24e","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 21:47:00.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/13/check/dsvm-tempest-kvm/566cad0 : FAILURE in 33m 38s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/13/check/dsvm-tempest-exe/32f43c2 : FAILURE in 34m 22s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"6a67cffebeb60456ac7496b6cd47468a81369da0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-08 21:57:10.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"d9c298af49e7eed549b4c08e8e5e657241b4d47f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-08 22:16:41.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/13/check/dsvm-tempest-infinibox-fc/9a000b8 : SUCCESS in 54m 38s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/13/check/dsvm-tempest-infinibox-iscsi/d9ab168 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":13},{"id":"2bddb31017e366424378488fb8a38682fc450d26","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-08 22:18:10.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/13/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f7f6499 : SUCCESS in 53m 14s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/13/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e9cc804 : SUCCESS in 1h 15m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":13},{"id":"4107cc996432515bafafffefab461bafba25ae97","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-08 22:36:17.000000000","message":"Patch Set 13:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/13/check/fujitsu-eternus-dx-iscsi/b353d97 : SUCCESS in 1h 33m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/13/check/fujitsu-eternus-dx-fc/eb8558c : FAILURE in 2m 23s","accounts_in_message":[],"_revision_number":13},{"id":"6a3392d3862c0abeed4f8362b23f1353ccfafa13","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-09 00:20:39.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/671195/13/check/dsvm-tempest-cisco-zonemanager-job/441b3ef/ : RETRY_LIMIT in 8m 58s","accounts_in_message":[],"_revision_number":13},{"id":"8b9b8fb77164b9b597211009afd644cb70d60a1b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-09 01:15:30.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/13/iscsi/ : FAILURE in 47s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/13/smb/ : FAILURE in 46s","accounts_in_message":[],"_revision_number":13},{"id":"e2e2bf45179144bc228fb466d55dbec179ebc24e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 03:00:10.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/95/671195/13/check/openstack-tox-lower-constraints/71f1aba/ : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://logs.opendev.org/95/671195/13/check/openstack-tox-pep8/28b4444/ : SUCCESS in 8m 16s\n- openstack-tox-py27 https://logs.opendev.org/95/671195/13/check/openstack-tox-py27/4ddcbb4/ : SUCCESS in 9m 21s\n- openstack-tox-py36 https://logs.opendev.org/95/671195/13/check/openstack-tox-py36/d0a4381/ : SUCCESS in 9m 03s\n- openstack-tox-py37 https://logs.opendev.org/95/671195/13/check/openstack-tox-py37/9c55afd/ : SUCCESS in 10m 45s\n- openstack-tox-docs https://logs.opendev.org/95/671195/13/check/openstack-tox-docs/0cf4d83/html/ : SUCCESS in 9m 16s\n- tempest-full https://logs.opendev.org/95/671195/13/check/tempest-full/8d28878/ : SUCCESS in 1h 27m 58s\n- neutron-grenade https://logs.opendev.org/95/671195/13/check/neutron-grenade/967bad8/ : SUCCESS in 1h 00m 11s\n- grenade-py3 https://logs.opendev.org/95/671195/13/check/grenade-py3/de2e2ef/ : SUCCESS in 1h 08m 19s\n- tempest-integrated-storage https://logs.opendev.org/95/671195/13/check/tempest-integrated-storage/cf6a880/ : SUCCESS in 1h 25m 56s\n- build-openstack-releasenotes https://logs.opendev.org/95/671195/13/check/build-openstack-releasenotes/90ebbad/html/ : SUCCESS in 13m 26s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/95/671195/13/check/cinder-tempest-dsvm-lvm-lio-barbican/df3bcb6/ : SUCCESS in 1h 26m 28s\n- cinder-tox-bandit-baseline https://logs.opendev.org/95/671195/13/check/cinder-tox-bandit-baseline/d8eeee8/ : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/95/671195/13/check/openstack-tox-functional-py36/78ca86e/ : SUCCESS in 6m 25s\n- cinder-rally-task https://logs.opendev.org/95/671195/13/check/cinder-rally-task/2103c9d/ : SUCCESS in 42m 21s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/95/671195/13/check/openstack-tox-pylint/bb74fd8/ : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/95/671195/13/check/cinder-plugin-ceph-tempest/ee1313a/ : SUCCESS in 1h 15m 10s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/95/671195/13/check/openstack-tox-functional/226c28d/ : SUCCESS in 5m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/95/671195/13/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/2ef39de/ : SUCCESS in 1h 08m 29s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/95/671195/13/check/legacy-tempest-dsvm-lvm-multibackend/fcd6cb5/ : SUCCESS in 1h 20m 31s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/95/671195/13/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/e82d595/ : SUCCESS in 1h 22m 05s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/95/671195/13/check/tempest-slow-py3/c57ce66/ : TIMED_OUT in 3h 06m 24s","accounts_in_message":[],"_revision_number":13},{"id":"f2c69707f5afc64c59b5bc91f46c19a19e5c7fca","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-09 03:06:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"406a75984ca84ad5ef9e7eee7bff0e2698a67afd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-09 03:26:15.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"c010331e51dd96b4c59eb1395d8df103dbb796a0","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-09 03:33:49.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1935 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2086 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1927 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"0a0216aaa027fb2059a9d861c2d8891cdd94ccc9","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 03:41:46.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/14/check/dsvm-tempest-kvm/83e64fc : FAILURE in 35m 07s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/14/check/dsvm-tempest-exe/ec1f28d : FAILURE in 45s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":14},{"id":"39850478cec74169b8abf038f21874f4e7ca2cc5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-09 03:57:27.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/14/iscsi/ : FAILURE in 1m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/14/smb/ : FAILURE in 50m 53s","accounts_in_message":[],"_revision_number":14},{"id":"6f5446e8f240135e2206b8604998d8dbe6e4b8c1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-09 04:20:01.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/14/check/dsvm-tempest-infinibox-fc/7fa1e59 : SUCCESS in 50m 38s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/14/check/dsvm-tempest-infinibox-iscsi/2b22086 : SUCCESS in 1h 13m 28s","accounts_in_message":[],"_revision_number":14},{"id":"71cb2d66be365ad592a426829eb8d2f2ec2f2cfb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-09 04:48:34.000000000","message":"Patch Set 14:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/14/check/fujitsu-eternus-dx-iscsi/f1f872d : SUCCESS in 1h 41m 50s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/14/check/fujitsu-eternus-dx-fc/227a600 : FAILURE in 2m 24s","accounts_in_message":[],"_revision_number":14},{"id":"0849e3f2aae62381f34f50a63fdf6c23fbe14963","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-09 04:54:02.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/671195/14/check/dsvm-tempest-cisco-zonemanager-job/0b01c40/ : RETRY_LIMIT in 8m 58s","accounts_in_message":[],"_revision_number":14},{"id":"8f2e4eea97e4e994be6416a14f4a231c4d71e9e4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-09 04:56:25.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/14/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f91c97c : SUCCESS in 57m 54s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/14/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/4dbf12a : FAILURE in 1h 50m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":14},{"id":"b2a95b148e3f626d777d157c697654a4e1ec27ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 05:30:39.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/95/671195/14/check/openstack-tox-lower-constraints/b0ed5e0/ : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://logs.opendev.org/95/671195/14/check/openstack-tox-pep8/af13b48/ : SUCCESS in 9m 01s\n- openstack-tox-py27 https://logs.opendev.org/95/671195/14/check/openstack-tox-py27/3db8636/ : SUCCESS in 10m 49s\n- openstack-tox-py36 https://logs.opendev.org/95/671195/14/check/openstack-tox-py36/9e71aa3/ : SUCCESS in 8m 13s\n- openstack-tox-py37 https://logs.opendev.org/95/671195/14/check/openstack-tox-py37/7335203/ : SUCCESS in 9m 51s\n- openstack-tox-docs https://logs.opendev.org/95/671195/14/check/openstack-tox-docs/aa1c60f/html/ : SUCCESS in 8m 25s\n- tempest-full https://logs.opendev.org/95/671195/14/check/tempest-full/8ef6e0c/ : SUCCESS in 1h 39m 20s\n- neutron-grenade https://logs.opendev.org/95/671195/14/check/neutron-grenade/1868832/ : SUCCESS in 58m 32s\n- grenade-py3 https://logs.opendev.org/95/671195/14/check/grenade-py3/2dce890/ : SUCCESS in 57m 45s\n- tempest-integrated-storage https://logs.opendev.org/95/671195/14/check/tempest-integrated-storage/ab22b61/ : SUCCESS in 1h 22m 14s\n- build-openstack-releasenotes https://logs.opendev.org/95/671195/14/check/build-openstack-releasenotes/b625288/html/ : SUCCESS in 15m 23s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/95/671195/14/check/cinder-tempest-dsvm-lvm-lio-barbican/979123c/ : SUCCESS in 1h 18m 03s\n- cinder-tox-bandit-baseline https://logs.opendev.org/95/671195/14/check/cinder-tox-bandit-baseline/9a30b87/ : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/95/671195/14/check/openstack-tox-functional-py36/e0f7748/ : SUCCESS in 4m 42s\n- cinder-rally-task https://logs.opendev.org/95/671195/14/check/cinder-rally-task/599d4af/ : SUCCESS in 53m 31s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/95/671195/14/check/openstack-tox-pylint/652f145/ : SUCCESS in 4m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/95/671195/14/check/cinder-plugin-ceph-tempest/4c88130/ : SUCCESS in 1h 10m 21s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/95/671195/14/check/openstack-tox-functional/76a3882/ : SUCCESS in 5m 31s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/95/671195/14/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/345414e/ : SUCCESS in 1h 05m 31s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/95/671195/14/check/legacy-tempest-dsvm-lvm-multibackend/c5a2562/ : SUCCESS in 1h 17m 02s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/95/671195/14/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/7a642d4/ : SUCCESS in 1h 30m 18s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/95/671195/14/check/tempest-slow-py3/b968227/ : SUCCESS in 2h 20m 25s","accounts_in_message":[],"_revision_number":14},{"id":"e5917d9bbc26960196a5c51e4462e1321e3596cf","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-09 06:12:35.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/95/671195/14/check/synology-dsm-cinder-iscsi/ffb3896 : FAILURE in 41m 23s","accounts_in_message":[],"_revision_number":14},{"id":"385c6ea55d64adf687efe81b8a80d725f9371a30","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-09 06:31:34.000000000","message":"Patch Set 14:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/14/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e2402de : FAILURE in 43m 30s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/14/check/3par-fc-driver-master-client-pip-ssa02-dsvm/5364930 : FAILURE in 42m 48s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/95/671195/14/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/2a75e48 : FAILURE in 42m 32s","accounts_in_message":[],"_revision_number":14},{"id":"2dc11a455e48b4ce489f6d9dec34b6cbbae5f94b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-09 08:23:08.000000000","message":"Patch Set 14:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-95-671195-14/ns5_iscsi : FAILURE in 74m 12s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-95-671195-14/ns5_nfs : FAILURE in 81m 19s","accounts_in_message":[],"_revision_number":14},{"id":"f96a11900a48f12b99bb565ea52b3043e77b3f5f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-09 09:16:21.000000000","message":"Patch Set 14:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-95-671195-14 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":14},{"id":"3b91bb126e90336d43bf16a46c1f98820ad5b417","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-09 12:01:33.000000000","message":"Patch Set 14:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/95/671195/14/check/kaminario-dsvm-tempest-full-iscsi/44e3184 : FAILURE in 5m 23s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/95/671195/14/check/kaminario-dsvm-tempest-full-FC/4484d0e : SUCCESS in 1h 28m 48s","accounts_in_message":[],"_revision_number":14},{"id":"5066e4a3136986d56d4e717d11a25d5083673424","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-09 15:23:47.000000000","message":"Patch Set 14:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/95/2489 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"014935a913e1bdf6571c86735b36b2b3ce1c2019","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-09 15:38:44.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/14/check/DellEMC_Unity_FC/10e5f37/DellEMC_Unity_FC/970 : SUCCESS in 1h 05m 31s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/14/check/DellEMC_Unity_iSCSI/f6d16a9/DellEMC_Unity_iSCSI/927 : SUCCESS in 1h 31m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"bff94730612c6833d1b71b0c89ee6028bef3ea96","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 16:18:18.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/14/check/DellEMC_XtremIO_FC/293cdd1/DellEMC_XtremIO_FC/1099 : SUCCESS in 1h 28m 02s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/14/check/DellEMC_XtremIO_iSCSI/d713059/DellEMC_XtremIO_iSCSI/1021 : SUCCESS in 1h 32m 16s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":14},{"id":"5bc8b204b4469358e92ecfba5ecc2ee49add2158","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 17:24:30.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/14/check/NEC-ISCSI/531d97b : SUCCESS in 2h 39m 40s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/14/check/NEC-FC/dd83f6d : SUCCESS in 2h 28m 21s","accounts_in_message":[],"_revision_number":14},{"id":"6c262e112dcc2079c397a2195e0b30d26ad43a7f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-09 17:29:25.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/95/671195/14/check/DellEMC_SC_FC/72fbc0b/DellEMC_SC_FC/421 : SUCCESS in 1h 08m 44s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/95/671195/14/check/DellEMC_SC_iSCSI/a9af89b/DellEMC_SC_iSCSI/402 : SUCCESS in 1h 49m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":14},{"id":"cfe6f1b86a90a6b067d02953666cbb910148dd2e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 18:52:35.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/95/671195/14/check/DellEMC_VxFlexOS/ab4aa40/DellEMC_VxFlexOS/542 : SUCCESS in 1h 52m 56s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":14},{"id":"d043473e1e696f8290846d4414b508ee6dee5d2b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 18:55:32.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/95/671195/14/check/DellEMC_VNX_iSCSI/584cffa/DellEMC_VNX_iSCSI/554 : FAILURE in 1h 47m 19s\n- DellEMC_VNX_FC http://publiclogs.emc.com/95/671195/14/check/DellEMC_VNX_FC/273652a/DellEMC_VNX_FC/565 : FAILURE in 1h 02m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":14},{"id":"5e0f24b417ec092b507da5cd42be87efb12cb448","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-09 20:31:31.000000000","message":"Patch Set 12:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"3daa27500c082c23fb6f0fc8f9899c9664cdd4b9","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-09 23:06:15.000000000","message":"Patch Set 13:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"561e33fd92f1ccbd16df43cca46b5c59e7eb9cb3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-10 01:05:39.000000000","message":"Patch Set 14:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/14/check/tempest-dsvm-full-xenial/67f6fe8/ : SUCCESS in 1h 26m 00s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/14/check/devstack-plugin-ceph-dsvm-xenial/7fc9fa2/ : SUCCESS in 1h 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d86fe803967d3cce7c778720be7ff4b06f441572","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-10 03:51:48.000000000","message":"Patch Set 14:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/671195-14/HPMSA-FC : SUCCESS in 46:36.94\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/671195-14/HPMSA-iSCSI : SUCCESS in 47:28.76","accounts_in_message":[],"_revision_number":14},{"id":"e9b1f52af7a957b1fdf9c3a091c165b51dd98de1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-10 05:01:38.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/95/671195/14/check-not-vote/ext-cinder-zuul/3fbd64b : FAILURE in 1m 48s","accounts_in_message":[],"_revision_number":14},{"id":"3fd9a911dfc8730b1d7d338ecd116509b74d982c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-10 07:20:25.000000000","message":"Patch Set 14:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"34022e9f5f82bf22227654f9b6321c8f6dfd47db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-10 20:49:29.000000000","message":"Patch Set 11:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/357/671195/11/check/infortrend-cinder-volume-fc/7df1c5d55b73aa72125d73a4338435aca1005f5c : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/400/671195/11/check/infortrend-cinder-volume-iscsi/7df1c5d55b73aa72125d73a4338435aca1005f5c : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"56df4e8484d8346a90b26afd02da708968021b36","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-11 03:30:17.000000000","message":"Patch Set 12:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/364/671195/12/check/infortrend-cinder-volume-fc/85f327d6789ea4d8f4558122f26f45345bfbb422 : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/407/671195/12/check/infortrend-cinder-volume-iscsi/85f327d6789ea4d8f4558122f26f45345bfbb422 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"d388f2239601833a27be081e743b6030908b785b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-11 04:23:00.000000000","message":"Patch Set 12:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/365/671195/12/check/infortrend-cinder-volume-fc/85f327d6789ea4d8f4558122f26f45345bfbb422 : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/408/671195/12/check/infortrend-cinder-volume-iscsi/85f327d6789ea4d8f4558122f26f45345bfbb422 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"8e8dee1968b789bcc46e4244adad34833590afbf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-11 05:16:30.000000000","message":"Patch Set 13:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/366/671195/13/check/infortrend-cinder-volume-fc/6be5bcd2535d86011bbfc379c45116c926115c56 : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/409/671195/13/check/infortrend-cinder-volume-iscsi/6be5bcd2535d86011bbfc379c45116c926115c56 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"20241dc9586b3fc18a6361eda959781812bb3121","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-11 06:10:20.000000000","message":"Patch Set 13:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/367/671195/13/check/infortrend-cinder-volume-fc/6be5bcd2535d86011bbfc379c45116c926115c56 : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/410/671195/13/check/infortrend-cinder-volume-iscsi/6be5bcd2535d86011bbfc379c45116c926115c56 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"eea8b7bf3adc1afb61b3e4a24cd0b41e3dedd2ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-11 12:29:04.000000000","message":"Patch Set 14:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/374/671195/14/check/infortrend-cinder-volume-fc/3f7c9889d78e594bcb3885e19b67a34e6aff1ffc : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/417/671195/14/check/infortrend-cinder-volume-iscsi/3f7c9889d78e594bcb3885e19b67a34e6aff1ffc : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"c2d7c3dc0d2cc3bd603998f15fcaf5663a648ec6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-11 13:24:23.000000000","message":"Patch Set 14:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/375/671195/14/check/infortrend-cinder-volume-fc/3f7c9889d78e594bcb3885e19b67a34e6aff1ffc : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/418/671195/14/check/infortrend-cinder-volume-iscsi/3f7c9889d78e594bcb3885e19b67a34e6aff1ffc : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"bf94f491aa143fdb44faa7f8ba18cba44e017d11","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-08-11 20:51:44.000000000","message":"Patch Set 14:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/671195-14/Lenovo-FC : FAILURE in 1:17:24\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/671195-14/Lenovo-iSCSI : FAILURE in 2:35:44","accounts_in_message":[],"_revision_number":14},{"id":"ab4e7e85618556b2de1f128d5e9bf9dc6b09da59","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-08-12 04:16:24.000000000","message":"Patch Set 14:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/671195-14/Seagate-FC : SUCCESS in 42:01.48\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/671195-14/Seagate-iSCSI : SUCCESS in 46:05.49","accounts_in_message":[],"_revision_number":14},{"id":"22151e7bd9938b931a7f1910b8259e29f83fb097","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-12 06:07:09.000000000","message":"Patch Set 14:\n\nrun-seagate-ci run-lenovo-ci run-hpmsa-ci","accounts_in_message":[],"_revision_number":14},{"id":"a731fefb7783a2c2fa76a564e0c49f07e6ce0d5c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-08-12 07:58:36.000000000","message":"Patch Set 14:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/671195-14/Seagate-FC : SUCCESS in 36:31.82\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/671195-14/Seagate-iSCSI : SUCCESS in 38:57.91","accounts_in_message":[],"_revision_number":14},{"id":"3d74b60aef135f053d9271ce21bb6aeba318a419","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-12 08:06:44.000000000","message":"Patch Set 14:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/671195-14/HPMSA-FC : SUCCESS in 42:04.60\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/671195-14/HPMSA-iSCSI : SUCCESS in 43:25.09","accounts_in_message":[],"_revision_number":14},{"id":"730b1fb86bc642657f871e9acfb96b1a1c683221","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-12 08:22:06.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1815 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1654 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"eaab9f1ad2690da109745ccfd4e7755ad4285108","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-13 17:46:45.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"22a61807ebed97a3cb7df9c7d5a478f647b61d38","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-13 17:47:26.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1824 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1621 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1663 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"2eca3b56de9e2eb5c8b6aa147fcdafe500c45de3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-13 18:48:13.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/15/iscsi/ : SUCCESS in 1h 00m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/15/smb/ : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":15},{"id":"175c10f279aa40059ac25b7a8c494de0f38663c4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-13 19:12:58.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/15/check/dsvm-tempest-kvm/db1b985 : FAILURE in 34m 54s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/15/check/dsvm-tempest-exe/b624551 : FAILURE in 35m 55s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"f26a9805e1604510652f86c647c921e7a6b5ec39","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-13 19:24:51.000000000","message":"Patch Set 15:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/15/check/fujitsu-eternus-dx-iscsi/e94d5b6 : SUCCESS in 1h 37m 26s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/15/check/fujitsu-eternus-dx-fc/efa86da : FAILURE in 2m 22s","accounts_in_message":[],"_revision_number":15},{"id":"74cc00d60738d619f3de67c6cc2ea6529b585b17","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-13 19:42:14.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/15/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1ec9bcd : SUCCESS in 1h 00m 53s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/15/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a37fec9 : FAILURE in 1h 49m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":15},{"id":"dff95cfde9ecbfe5828d4501574e9d81dd994616","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-13 21:01:35.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/671195/15/check/dsvm-tempest-cisco-zonemanager-job/82b2ef1/ : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":15},{"id":"80610091c2db1c69e82744a2f0af1b0219af7ba9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 21:14:29.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86b30715927148f0864b75fbcd3aa30c : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8f6c6f83574e07b756b80ffec60564 : SUCCESS in 9m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5acb36a3eab24088ad177597f8f8378a : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4efec4b4eec4093b1f3f5da95c4bcca : SUCCESS in 9m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85d9b430b89d472688050907315b6bcf : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2eed09b13804d1394e2227dca920ee9 : SUCCESS in 7m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a4cf95272d14e828145a460295b6873 : SUCCESS in 1h 58m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/54617c495c0b40ed93bdb813d956d66e : SUCCESS in 1h 09m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7a9197e1e6de452c83c53e30ee3d84bb : SUCCESS in 1h 07m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8dc79148c0da4821bf0edd982748c125 : SUCCESS in 2h 00m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/522b5cee07e84fbda1a839740998c57f : SUCCESS in 14m 55s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8909efa756c742ee88c7f7f4cfe2ae17 : SUCCESS in 1h 26m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/48e307c323384825863afbd390b0194d : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ea716c5c33645c68ad9aab816130eff : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6818d3bfa4ea4ecaa03ea959f233af4e : SUCCESS in 40m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f10d0556d634bd8822ec230d13f0529 : SUCCESS in 4m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b99bb12dfcd4507bbaf7ad6831d8378 : SUCCESS in 1h 21m 13s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d6a5a63f5be4a909703b36f11b46236 : SUCCESS in 7m 22s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/354add2ed7be4f73a306fd8b0cce9f35 : SUCCESS in 1h 12m 59s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b24b63486bf4dfd8d739cce5404484f : SUCCESS in 1h 40m 43s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/23b5461b2d3a4484ad835494cd40e0fc : SUCCESS in 1h 43m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3cc47b2798c643759b64980cb9d1871f : SUCCESS in 2h 08m 47s","accounts_in_message":[],"_revision_number":15},{"id":"fe82d13f98f4ebe3abda1ae44ced3f4c867a02bc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-13 22:14:47.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/95/671195/15/upstream-check/cinder-cDOT-iSCSI/2bc6979/ : SUCCESS in 50m 17s\n- cinder-cDOT-NFS http://13.56.159.105/logs/95/671195/15/upstream-check/cinder-cDOT-NFS/3094466/ : SUCCESS in 52m 21s\n- cinder-cDOT-FCP http://13.56.159.105/logs/95/671195/15/upstream-check/cinder-cDOT-FCP/fa18ca8/ : SUCCESS in 59m 30s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"c0b70c3921c7b573aa4e8931bdc1e58afbd043ae","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-13 22:17:53.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/95/671195/15/upstream-solidfire-check/cinder-solidfire-iSCSI/1809b66/ : SUCCESS in 1h 02m 06s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"1e8646c7e568ae7d33693ee97e6f91d68b942fe0","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-13 22:19:13.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/95/671195/15/check/synology-dsm-cinder-iscsi/62f8012 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":15},{"id":"999d8190f62be5ab7902e5c46ffe7fc33a1c7b92","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-08-13 23:06:19.000000000","message":"Patch Set 15:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/671195-15/Seagate-FC : SUCCESS in 37:04.37\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/671195-15/Seagate-iSCSI : SUCCESS in 40:07.52","accounts_in_message":[],"_revision_number":15},{"id":"6cf76df0676985729a10ef02ff642a455cee5df3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-13 23:28:13.000000000","message":"Patch Set 15:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d6e5031 : SUCCESS in 2h 05m 17s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/15/check/3par-fc-driver-master-client-pip-ssa02-dsvm/7d754f5 : SUCCESS in 2h 11m 10s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/95/671195/15/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/6aec0c0 : FAILURE in 1h 27m 37s","accounts_in_message":[],"_revision_number":15},{"id":"9e73e122320484790e315dbb40b2e44ec13a0596","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-13 23:41:06.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/95/671195/15/check/DellEMC_SC_FC/d00db93/DellEMC_SC_FC/447 : SUCCESS in 1h 02m 44s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/95/671195/15/check/DellEMC_SC_iSCSI/16df73c/DellEMC_SC_iSCSI/428 : SUCCESS in 1h 43m 48s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":15},{"id":"bfc5828e95ab629bd1b5aa6e2665ebf6a738b0d4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-14 00:02:55.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/95/671195/15/check/DellEMC_VxFlexOS/b726d2e/DellEMC_VxFlexOS/568 : SUCCESS in 1h 53m 04s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":15},{"id":"ec0600d9f884d702efe4e1c9406b2c4050f83926","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-14 00:16:14.000000000","message":"Patch Set 15:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/671195-15/HPMSA-FC : SUCCESS in 43:22.94\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/671195-15/HPMSA-iSCSI : SUCCESS in 45:40.97","accounts_in_message":[],"_revision_number":15},{"id":"688bee9869570a5cca4d702e4ee461a87c438c86","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-14 00:19:10.000000000","message":"Patch Set 15:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/95/671195/15/check/kaminario-dsvm-tempest-full-iscsi/4d00869 : SUCCESS in 1h 39m 43s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/95/671195/15/check/kaminario-dsvm-tempest-full-FC/a0cd895 : FAILURE in 5m 15s","accounts_in_message":[],"_revision_number":15},{"id":"5b878892fd817ff759a751b3d531c7e8c444cf61","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-14 00:34:13.000000000","message":"Patch Set 15:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-95-671195-15 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":15},{"id":"fe4ef5ab702ac2e03e1da6f885e1d044a064cd8b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-14 00:39:55.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/15/check/DellEMC_XtremIO_FC/51f661e/DellEMC_XtremIO_FC/1141 : SUCCESS in 1h 30m 11s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/15/check/DellEMC_XtremIO_iSCSI/fc0eddd/DellEMC_XtremIO_iSCSI/1063 : SUCCESS in 1h 33m 15s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"a43c921a635165360c11d044812784e9cda81977","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-14 00:50:18.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/15/check/DellEMC_Unity_FC/b0b86f5/DellEMC_Unity_FC/1015 : SUCCESS in 1h 12m 40s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/15/check/DellEMC_Unity_iSCSI/14302ae/DellEMC_Unity_iSCSI/969 : SUCCESS in 1h 24m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"8717afdcbed26a3b00cd3862376258de73f3afc0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-14 02:03:02.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/95/671195/15/check/DellEMC_VNX_iSCSI/712ef9c/DellEMC_VNX_iSCSI/581 : FAILURE in 2h 53m 58s\n- DellEMC_VNX_FC http://publiclogs.emc.com/95/671195/15/check/DellEMC_VNX_FC/224c80d/DellEMC_VNX_FC/592 : FAILURE in 8m 03s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":15},{"id":"90c7bd327b6d135cae1c41bc1b05b5f8a0502812","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-14 02:19:57.000000000","message":"Patch Set 15:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/449/671195/15/check/infortrend-cinder-volume-fc/1e456ea750ea4b28e073ecd35db15c0fec8de918 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"d843ac8ed6f401b598852fa6cecbdba07b0a267e","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-14 02:35:48.000000000","message":"Patch Set 15:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/95/2522 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"85ae6d6178043ce65bd0e807869c3f2792567a80","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-14 03:26:03.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/95/671195/15/check-cinder/Cinder-ISER-ISCSI/d037c09 : SUCCESS in 1h 10m 03s\n- Cinder-ISER-LIO http://13.74.249.42/95/671195/15/check-cinder/Cinder-ISER-LIO/dbaff04 : SUCCESS in 1h 03m 40s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"35fdea7cabc908c0a4f8711c751c1959d93eda1f","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-08-14 03:48:09.000000000","message":"Patch Set 15:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/95/671195/15/check/Cinder-AS13000/49f3ba8 : SUCCESS in 2h 26m 52s\n- cinder-test-fc http://218.29.111.78:9000/95/671195/15/check/cinder-test-fc/d9dabf7 : SUCCESS in 56m 10s\n- cinder-test-iscsi http://218.29.111.78:9000/95/671195/15/check/cinder-test-iscsi/bcd624c : FAILURE in 33m 01s","accounts_in_message":[],"_revision_number":15},{"id":"32e39fe72b175bf47b93ddd1c27cc51633944539","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-14 06:11:18.000000000","message":"Patch Set 15:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/15 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"6ede00dff03c92b10d72d3cbe91bf3ad4574b6e3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-14 06:16:59.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/15/check/NEC-ISCSI/7c92fa2 : SUCCESS in 1h 54m 16s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/15/check/NEC-FC/c382c5a : SUCCESS in 2h 08m 28s","accounts_in_message":[],"_revision_number":15},{"id":"21459f9a3b7822af53b30913da5e6825f304c665","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-14 08:43:15.000000000","message":"Patch Set 15:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/444/671195/15/check/infortrend-cinder-volume-fc/1e456ea750ea4b28e073ecd35db15c0fec8de918 : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/466/671195/15/check/infortrend-cinder-volume-iscsi/1e456ea750ea4b28e073ecd35db15c0fec8de918 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"0d61aaf91a3618079cd306c91984feaa7051f24e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-14 11:49:45.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/15/check/dsvm-tempest-infinibox-fc/f1c53e0 : SUCCESS in 56m 09s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/15/check/dsvm-tempest-infinibox-iscsi/c9c0200 : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":15},{"id":"dcbbbc17f1da32aa0581aeb7121d084298426dce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 14:04:18.000000000","message":"Patch Set 15: Code-Review-1\n\n(3 comments)\n\nOne minor issue with the current code.","accounts_in_message":[],"_revision_number":15},{"id":"aa7c7b2fc8cfd539a18ac1f8e0fe4518e0d79369","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-08-14 14:05:15.000000000","message":"Patch Set 15: Code-Review-1\n\nHPMSA CI is missing logs","accounts_in_message":[],"_revision_number":15},{"id":"326d03e5d6b28c48e3a3dbc9653542a0f6a89ad2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 14:07:43.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"365109c1a11555e6c15a0db56d110b1f1d95479c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 14:09:39.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"002567353782d769a65b5926481241af6d90cccb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 14:10:45.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"52efd3a32da1c350eb38d602e37b135c13ba886e","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-16 03:07:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"6f03fe6f80116786f6eb28bcaf482623d4e264a2","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-16 03:33:07.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1681 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1646 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1850 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"30a5eca7021fe0a18f66d1bb063544a43b70f239","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 03:55:27.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/16/check/dsvm-tempest-kvm/06465b8 : FAILURE in 38m 06s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/16/check/dsvm-tempest-exe/fa90c8e : FAILURE in 38m 05s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"ed3b90d8e59a7509727110b1826c0be3033478d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-16 04:05:47.000000000","message":"Patch Set 16:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/496/671195/16/check/infortrend-cinder-volume-iscsi/4056dd658ade1bdfa67cf82b2e80d43ef0ce4b52 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/477/671195/16/check/infortrend-cinder-volume-fc/4056dd658ade1bdfa67cf82b2e80d43ef0ce4b52 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"20d3cfbd0f22cb8984f492768420cc67ee5b16ef","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-16 04:11:01.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/16/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/0c30cc2 : FAILURE in 6m 47s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/16/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/3db985e : SUCCESS in 1h 03m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":16},{"id":"55d52a77a4a64921963c6ba7d6d67547e04ea403","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-16 04:18:23.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/16/check/dsvm-tempest-infinibox-fc/3a4eda8 : SUCCESS in 1h 09m 53s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/16/check/dsvm-tempest-infinibox-iscsi/8fb085b : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":16},{"id":"e20a1b7e73b483696798633931d2ae637f0e25a6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 04:37:45.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/16/check/DellEMC_XtremIO_FC/822680a/DellEMC_XtremIO_FC/1163 : SUCCESS in 1h 29m 55s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/16/check/DellEMC_XtremIO_iSCSI/0ecf2ea/DellEMC_XtremIO_iSCSI/1085 : SUCCESS in 1h 29m 18s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":16},{"id":"cb352a2aa5e6550ecb9964e9c64e7cbdf7734bda","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-16 04:38:22.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/16/iscsi/ : FAILURE in 1h 30m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/16/smb/ : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":16},{"id":"97119cf29f6b3c1c78268f922eb3ac06eb726cbd","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-16 04:38:24.000000000","message":"Patch Set 16:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/16/check/fujitsu-eternus-dx-iscsi/b57aa33 : SUCCESS in 1h 30m 13s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/16/check/fujitsu-eternus-dx-fc/f6e48cb : FAILURE in 53s","accounts_in_message":[],"_revision_number":16},{"id":"1d8c355a1a048f689707a703b3d471d38e36374c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 05:20:42.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/16/check/NEC-ISCSI/7bce9d5 : SUCCESS in 2h 09m 13s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/16/check/NEC-FC/75568cd : SUCCESS in 2h 12m 46s","accounts_in_message":[],"_revision_number":16},{"id":"55a4cd564db9ba7dc4e8bf049dab9ed4bb975935","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-16 05:33:30.000000000","message":"Patch Set 16:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/497/671195/16/check/infortrend-cinder-volume-iscsi/4056dd658ade1bdfa67cf82b2e80d43ef0ce4b52 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/478/671195/16/check/infortrend-cinder-volume-fc/4056dd658ade1bdfa67cf82b2e80d43ef0ce4b52 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"06ef8d2105fc5f067886c2b6f114afc5fe73ffba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 05:43:08.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28636fc2f5e84bd2bedfa49ca25c0928 : FAILURE in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51335e173934ac084aa0ed50fee77bd : FAILURE in 9m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e13faa8d1654f008e0bedb7c4ba295c : FAILURE in 13m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49d67ba6fd404d2b900b61bcdaeb0b93 : FAILURE in 12m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5af78d580414f739eb27c11d1714fa8 : FAILURE in 18m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861e2eec2e834bf49f32d233f5b0f7cc : SUCCESS in 11m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f2aa4ac09a314363aa7f156a720861cd : SUCCESS in 1h 40m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/09061805c2314f509c79675910ef41d0 : SUCCESS in 57m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eef375cc6bb94e60a94a4c89bd328cb3 : SUCCESS in 1h 05m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e317f3eb256c4dd9af4fca0c8bb07c48 : SUCCESS in 1h 28m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8787c25ba3454742a43d004841aef2ef : FAILURE in 16m 54s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/49947310978640a2b5a61ccdda37bd46 : SUCCESS in 1h 32m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a939c4ce0ac4080914c1b9f0a33490e : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0b36dc8abb04229a47b620846a71020 : SUCCESS in 8m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a288b2d032c45078c9ffe663681013d : SUCCESS in 41m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5013bf77a33f49b4a3d83e7a1c4423cf : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d69740bbdbeb49d1b3fa0aed01a5d040 : SUCCESS in 1h 26m 24s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/27f94f5409034fa8b655fa0467a9ccf1 : SUCCESS in 7m 30s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c770823df8eb44b5bba82e7ba3a4cbe6 : SUCCESS in 1h 08m 55s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/296e3dd7199f4801bc2779eebec0975c : FAILURE in 1h 19m 12s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/afe9ed195a434a81a33275b7ac4aa8d7 : SUCCESS in 1h 31m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/487f8b1f0f7a4e89aafc6f4fe825e79a : SUCCESS in 2h 24m 54s","accounts_in_message":[],"_revision_number":16},{"id":"4e46a0e5a5750097064f51ef82a8e4cd4a508689","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-16 06:11:58.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"6a438138cccb97155f07e508b82e4d77ecc42ac7","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-16 06:23:06.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1647 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1682 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1851 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"f4d38b82e8fb18025126ca4db54ef420acb835c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-16 07:08:57.000000000","message":"Patch Set 17:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/498/671195/17/check/infortrend-cinder-volume-iscsi/7c5d12440a35aac7c1fd398659c58b42dc9e8447 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/479/671195/17/check/infortrend-cinder-volume-fc/7c5d12440a35aac7c1fd398659c58b42dc9e8447 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"e7e0762998ceaa76c40a86f93e4a9f2782605511","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-16 07:18:22.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/17/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/e774b1e : FAILURE in 6m 43s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/17/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e9f9a0a : SUCCESS in 1h 05m 58s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":17},{"id":"a46d416f66405fd2c5260f490c2c6832d1842147","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-16 07:19:45.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/17/check/dsvm-tempest-infinibox-fc/853fccf : SUCCESS in 55m 49s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/17/check/dsvm-tempest-infinibox-iscsi/e22914e : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":17},{"id":"b8631c998833df4d641fbeb67d8f1a5577a52a3f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-16 07:22:18.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/17/iscsi/ : SUCCESS in 59m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/17/smb/ : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":17},{"id":"e503c014a1a93e3c0e3163818e02707a71c9e749","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 07:41:01.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/17/check/DellEMC_XtremIO_FC/ff0ac70/DellEMC_XtremIO_FC/1164 : SUCCESS in 1h 28m 46s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/17/check/DellEMC_XtremIO_iSCSI/c0ae469/DellEMC_XtremIO_iSCSI/1086 : FAILURE in 1h 28m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":17},{"id":"0fdee10bce7bd0ba47fa0d9aa971acec2937764d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 08:00:21.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/17/check/dsvm-tempest-kvm/0d875fb : FAILURE in 1h 23m 35s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/17/check/dsvm-tempest-exe/bb2a886 : FAILURE in 1h 29m 07s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"6f1500944cc16f9a59b5c4a4efd1dc1cc2f07231","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 08:30:25.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/17/check/NEC-ISCSI/e2bb921 : SUCCESS in 2h 13m 39s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/17/check/NEC-FC/c535b98 : SUCCESS in 2h 17m 05s","accounts_in_message":[],"_revision_number":17},{"id":"ed490cd4f1487ea3a069e22dd275b8111a216598","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:14:08.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f12009b8839f400fbda30e1943a65e29 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4598cf4757ae448b9f0ada0e3d6940fa : FAILURE in 9m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9007b3c88d274b5980904eb4bf4fdbb2 : SUCCESS in 10m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34fc7db289b499184179d31f548bca5 : SUCCESS in 9m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c2dc879a1f24bf9a04abfc17fa8911d : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d287cab3c0864417983beabf57b47a6d : SUCCESS in 9m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a40909406ac341d7a3ab37de3bfee5c9 : SUCCESS in 1h 35m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7809e6aee4cd4f6888d6a7e0c2c9d5f9 : SUCCESS in 1h 03m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a37cc6f41b4c408eafdb3a3bed62075c : SUCCESS in 57m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f862e7dc6de74c25b56ae6c14d961e96 : SUCCESS in 1h 18m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6a3153329cb4328a5bd4b0d8cf6b2b0 : SUCCESS in 13m 13s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6da8349973f24d10827efa985923153b : SUCCESS in 1h 23m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d4ffbd505b4347e0bfac5ddd643e7d1e : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bfc0eddc14d94cc1afd4fd81731c80db : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6567dc693461474881afb978fe1cb39d : SUCCESS in 43m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a54085bdef5442e8aa60eb9011edf725 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4e321beade824303a0b7bed4ce35e920 : SUCCESS in 1h 09m 59s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/87077a1ccf6d4b7b95cc0882fb97ef37 : SUCCESS in 7m 00s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f89bef281d0946ecb7f505ed8008066a : SUCCESS in 1h 13m 52s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/072f86f50bdc4e319bde4dc2f653039a : SUCCESS in 1h 39m 06s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/112dad5383d541dc9a2141e413249444 : SUCCESS in 1h 35m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f25c7e169be643958d1885199050e00e : SUCCESS in 2h 29m 56s","accounts_in_message":[],"_revision_number":17},{"id":"77a5cc8f3783103738d868de3d4cc627bcc3e2e7","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-16 14:37:31.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9040f930e0cdf7a2f5788f5af8728eec05080e96","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-16 15:04:31.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1685 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1650 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1854 : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"f41ff930cb94098b9911513c5f62cf35e62eb510","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 15:20:46.000000000","message":"Patch Set 18:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/18/check/NEC-ISCSI/d0ac3f3 : FAILURE in 40m 18s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/18/check/NEC-FC/752b5fc : FAILURE in 40m 25s","accounts_in_message":[],"_revision_number":18},{"id":"3a2aab8f22b0dd3c3c3409f13021e8bd397d0231","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 15:21:58.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/18/check/dsvm-tempest-kvm/8ef4522 : FAILURE in 36m 35s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/18/check/dsvm-tempest-exe/7109101 : FAILURE in 34m 50s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"7550eb9e2586541bcac503a96188719b594c9ad3","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-16 15:42:53.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/18/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/bf29282 : FAILURE in 7m 10s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/18/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/bb062f5 : SUCCESS in 1h 04m 57s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":18},{"id":"a652279b6fb3910aa4068c5f7564d2eaeecc76ff","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-16 15:48:52.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/18/iscsi/ : SUCCESS in 1h 10m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/18/smb/ : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":18},{"id":"2bc2a6d1f50667e781adc90103e4c387429d70ad","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-16 16:10:05.000000000","message":"Patch Set 18:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/18/check/fujitsu-eternus-dx-iscsi/c3532f8 : SUCCESS in 1h 31m 53s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/18/check/fujitsu-eternus-dx-fc/d5d827c : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":18},{"id":"bb3c82a0bbcd7e4909b8486af186f43e860063a1","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-16 16:30:00.000000000","message":"Patch Set 16:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"6e953c0887e39ab904cbe08469fa36c71ac84be4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-16 16:34:45.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/18/check/DellEMC_Unity_FC/d1ef999/DellEMC_Unity_FC/1034 : SUCCESS in 1h 10m 13s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/18/check/DellEMC_Unity_iSCSI/3a4e80c/DellEMC_Unity_iSCSI/986 : SUCCESS in 1h 28m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"b9171501d97c86d626bced7ebc8fcfe92da2eac2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 17:01:00.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/18/check/DellEMC_XtremIO_FC/7a42b3a/DellEMC_XtremIO_FC/1168 : SUCCESS in 1h 28m 05s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/18/check/DellEMC_XtremIO_iSCSI/80772b0/DellEMC_XtremIO_iSCSI/1090 : SUCCESS in 1h 31m 49s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":18},{"id":"6588d12d9e23534bbbeb9bdaeae9afafed09a7bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 17:17:30.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9293ab8e029c469da0e0229e8992d3b4 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abde21a18a074bcea3af1d54f25e6dfe : FAILURE in 10m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e8af850afb340ec91f13ca3cc62ed61 : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67a2d94e2b5d460ea768ef8c4cd4ba7e : SUCCESS in 10m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3fb7fd2e71f48419006650f74f8ecd3 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1db0eb2b1db4c5abd863cece0c00a48 : SUCCESS in 8m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7110e50158c64ded95f3e4d96926cbbb : SUCCESS in 1h 43m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f6d4d8abdf594e6a82adcab30e875e9b : SUCCESS in 1h 03m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/105c6fafca084c6b8061b29849b77b55 : SUCCESS in 56m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30f4001698c64fc4bb6e1175abbd184c : SUCCESS in 1h 28m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eace560fbb04f238f9f556e4d7ce3b1 : SUCCESS in 15m 35s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9daeeb5dfd384efd8a436add272bc665 : SUCCESS in 1h 33m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/39a986ea559f4c2d9cbeee1bd870862c : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c934eb1fb8ec40d29aa2e6e4a91efb2c : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8edd39847d6f4feeac680880aa7c11f6 : SUCCESS in 42m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/12ba50dfd420430a87beb2d9b1ea3ab6 : SUCCESS in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4c5df1dd299f4266a65e8c2e87342d80 : SUCCESS in 1h 21m 25s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ef535f2e0ad2446f8497529a57a2173e : SUCCESS in 5m 36s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/931c08b5e4e844249507a7d984d007e7 : SUCCESS in 1h 11m 53s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fd152fd59833466a98ee17454a8b1c60 : SUCCESS in 1h 32m 58s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/a3dcd6a2338241c385ef2e7812f61f7c : SUCCESS in 1h 32m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/186580153c2f47a0b78e58223fe5f23f : SUCCESS in 2h 38m 22s","accounts_in_message":[],"_revision_number":18},{"id":"e15bbccb456f16e9acd50831140dcbff22c0f417","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-16 17:25:16.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"2b5209673f949add222017c5c4a3cacd61e09fd9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-16 17:26:04.000000000","message":"Patch Set 19:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/19/check/dsvm-tempest-infinibox-fc/5cb0377 : NOT_REGISTERED\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/19/check/dsvm-tempest-infinibox-iscsi/5f90e17 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":19},{"id":"9f8729d2ddbe0eb0ff63d4cdde2aa729142c4208","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-16 17:44:44.000000000","message":"Patch Set 17:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"4021674feac0de0843f7e6c728533585e0be5329","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-16 17:51:30.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1686 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1651 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1855 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"9b4aed5e6684f1abd9f5485fafe7553c04f82688","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-16 18:05:27.000000000","message":"Patch Set 19:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":19},{"id":"1b8c3e132ddf97a093cafc07d8e290e8447de3f7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-16 18:05:30.000000000","message":"Patch Set 18:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":18},{"id":"b34239fbec6804f6fa3a464f6e1008ccab2a07eb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-16 18:05:50.000000000","message":"Patch Set 17:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":17},{"id":"f64f25820a3a8225aefba12b3f9750c0086ec153","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-16 18:05:53.000000000","message":"Patch Set 16:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":16},{"id":"2a417b4bc8c700b5b53fda96f924972eb8d1d594","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 18:10:19.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/19/check/dsvm-tempest-kvm/98fb84c : FAILURE in 34m 50s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/19/check/dsvm-tempest-exe/acb3a91 : FAILURE in 34m 31s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":19},{"id":"f92fbcff1dbea7787ddaedbf89781f7176a24449","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-16 18:35:19.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/19/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/2892021 : FAILURE in 20s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/19/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/3fa1dfa : SUCCESS in 1h 09m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":19},{"id":"66ad09c329b0f1269bfb8b538c45a12f13d59972","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-16 18:49:16.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/19/check/DellEMC_Unity_FC/0dd97c2/DellEMC_Unity_FC/1035 : SUCCESS in 1h 11m 02s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/19/check/DellEMC_Unity_iSCSI/45f4065/DellEMC_Unity_iSCSI/987 : SUCCESS in 1h 23m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"3628b67d1faca13b72f31d3be7a45919abb7e69b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-16 18:50:14.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/19/iscsi/ : FAILURE in 1h 24m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/19/smb/ : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":19},{"id":"b271591da4eef3a2fc39b48f1582cb441e431046","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 18:56:55.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/19/check/DellEMC_XtremIO_FC/221d427/DellEMC_XtremIO_FC/1169 : SUCCESS in 1h 27m 57s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/19/check/DellEMC_XtremIO_iSCSI/8b3c515/DellEMC_XtremIO_iSCSI/1091 : SUCCESS in 1h 31m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":19},{"id":"88493eba79fa54dbcc3c63201fa81dc01f5819d3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-16 19:01:13.000000000","message":"Patch Set 19:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/95/671195/19/check/fujitsu-eternus-dx-iscsi/ce11f35 : SUCCESS in 1h 35m 14s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/95/671195/19/check/fujitsu-eternus-dx-fc/7ab93ab : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":19},{"id":"25a5acb2bf32f03f6f41b89a1af5db02b23775ed","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 19:39:08.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/19/check/NEC-ISCSI/31282df : SUCCESS in 1h 58m 51s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/19/check/NEC-FC/ef1cbd2 : SUCCESS in 2h 12m 39s","accounts_in_message":[],"_revision_number":19},{"id":"ba656e8a262ec208e6ce82269f20adfcf82b69e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 19:57:08.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd2b70a65a44448c9e26ce137d4dd16a : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a100de95ae424472b9f3d47abacf0109 : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bb550f025c943fe8fef7bf5e24d8c6d : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/684213d70f134577823423ab4505bd18 : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5ffda048ed7418b8b4eb55f12b09790 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c8918c1ab646ebaf7f39b2e9a06f81 : SUCCESS in 9m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/154f97bedfa44dfb8a58fb94aff4d9b7 : SUCCESS in 1h 39m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e471d9e2e14944358938c8c71b33d13b : SUCCESS in 54m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/14cdfaad07624993af11aa020706284d : SUCCESS in 1h 04m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55727c6b78f14bae886e4c5fdf8a2b82 : SUCCESS in 1h 23m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04418d58673e40009aaf93b3e1c82a3e : SUCCESS in 15m 13s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8ef97193b4004cf39808dc0b4f6dff09 : SUCCESS in 1h 32m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e81dd796a7ed420c86de3b7acb58484d : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4dc1332fe824876a06a48f64aebd504 : SUCCESS in 4m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be55d5f97759469f995b01b05a482497 : SUCCESS in 37m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5e6aeed16184bf59b8da5b5ea5c2622 : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d3a69650e22146e1ba339ac1ec338c88 : SUCCESS in 1h 11m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fbb9c1f970f34396adf218bb1ab8d96d : SUCCESS in 5m 30s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/858d0049f6d74e3b88eabb0370e4e3cc : SUCCESS in 1h 07m 08s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dcbd6a0bf78141a2a32a478868cabe8b : FAILURE in 1h 34m 30s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/fafa99503f7a4cb2a8252595946612d1 : SUCCESS in 1h 35m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/300eb7f928bf4d7f99158b9d0adaeecf : SUCCESS in 2h 21m 41s","accounts_in_message":[],"_revision_number":19},{"id":"a482717d6d461648e3dd67ff9c8fc68a091326be","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-16 20:37:29.000000000","message":"Patch Set 19:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-95-671195-19/ns5_iscsi : SUCCESS in 38m 32s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-95-671195-19/ns5_nfs : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":19},{"id":"c2feb1c0cef49a989c1302b66e5f4026bab86eed","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-16 21:02:14.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/95/671195/19/check/synology-dsm-cinder-iscsi/58f7a45 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":19},{"id":"e61110f5b4eb340b5c47ca0fe012ef6cfb626f21","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-16 21:09:26.000000000","message":"Patch Set 19:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/95/2544 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"d06749a3b24676baa6d6d33241051f0a7b7cf622","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-16 21:21:55.000000000","message":"Patch Set 19:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-95-671195-19 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":19},{"id":"dfb0fff663bdee16cff6d30b50f70d27d60d12b8","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-16 21:25:39.000000000","message":"Patch Set 19:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/671195-19/HPMSA-FC : FAILURE in 9:53.26\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/671195-19/HPMSA-iSCSI : FAILURE in 10:54.70","accounts_in_message":[],"_revision_number":19},{"id":"1ff47ef79216520e7c87f34520eea5c99efbef3c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-08-16 21:31:11.000000000","message":"Patch Set 19:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/671195-19/Seagate-FC : SUCCESS in 34:47.20\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/671195-19/Seagate-iSCSI : SUCCESS in 37:43.33","accounts_in_message":[],"_revision_number":19},{"id":"4055d698363060ca3ce9ed976d2920b51ef097e2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 21:33:21.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/19/check/DellEMC_XtremIO_FC/63491fb/DellEMC_XtremIO_FC/1170 : FAILURE in 1h 28m 01s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/19/check/DellEMC_XtremIO_iSCSI/1a6f599/DellEMC_XtremIO_iSCSI/1092 : SUCCESS in 1h 35m 54s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":19},{"id":"ec68e71f3b02211991a404cd0f3f3ba8074ef4db","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-16 21:38:56.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/19/check/DellEMC_Unity_FC/a6714a0/DellEMC_Unity_FC/1036 : SUCCESS in 1h 15m 56s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/19/check/DellEMC_Unity_iSCSI/afe73db/DellEMC_Unity_iSCSI/988 : SUCCESS in 1h 41m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"7eab02d218c84b23dcadfb428c20ccfebc021216","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 21:49:03.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/95/671195/19/check/DellEMC_VNX_iSCSI/c94b8c5/DellEMC_VNX_iSCSI/598 : SUCCESS in 1h 51m 37s\n- DellEMC_VNX_FC http://publiclogs.emc.com/95/671195/19/check/DellEMC_VNX_FC/a493676/DellEMC_VNX_FC/610 : FAILURE in 9m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":19},{"id":"d0cb460b228eb8704b0c7f53da0c390303e81f8d","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-16 21:49:16.000000000","message":"Patch Set 19:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/95/671195/19/check/kaminario-dsvm-tempest-full-iscsi/0003d3d : SUCCESS in 1h 51m 39s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/95/671195/19/check/kaminario-dsvm-tempest-full-FC/cdcebab : FAILURE in 2m 14s","accounts_in_message":[],"_revision_number":19},{"id":"3a9bd4462ab060283563d656d3dc10a082875096","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 21:52:49.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/95/671195/19/check/DellEMC_VxFlexOS/d6462b2/DellEMC_VxFlexOS/585 : SUCCESS in 1h 54m 34s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":19},{"id":"c559d633b2c00921abe2d400d1433049c36f2d57","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-16 22:03:48.000000000","message":"Patch Set 19:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/19/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/bedf81a : SUCCESS in 1h 58m 30s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/19/check/3par-fc-driver-master-client-pip-ssa02-dsvm/ed53d8e : SUCCESS in 2h 04m 04s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/95/671195/19/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/9b087f6 : FAILURE in 1h 32m 41s","accounts_in_message":[],"_revision_number":19},{"id":"1552ee2dfc3cdf0033bac906b9cd64e84fb2485d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-16 22:49:44.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/95/671195/19/check/DellEMC_SC_FC/a91ad1a/DellEMC_SC_FC/465 : SUCCESS in 1h 06m 56s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/95/671195/19/check/DellEMC_SC_iSCSI/8b4f2da/DellEMC_SC_iSCSI/445 : FAILURE in 2h 52m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":19},{"id":"69ee1b8758004f6666c9353171a65f8f33ce7cd9","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-16 23:50:24.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"47d10fd7de9709e03ddc3ce6f658d791bcf14a45","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-16 23:50:47.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"c6597c03e9c6820a75d4f2efcd834f37a6179230","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-17 00:12:42.000000000","message":"Patch Set 20:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/95/671195/20/check/dsvm-tempest-infinibox-fc/baabe0f : FAILURE in 16m 08s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/95/671195/20/check/dsvm-tempest-infinibox-iscsi/3f6d948 : FAILURE in 21m 42s","accounts_in_message":[],"_revision_number":20},{"id":"d043cdc8c43f8f0b4f828b3868daca1515900ce5","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-17 00:17:07.000000000","message":"Patch Set 20:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1687 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1856 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1652 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"df6db1c28f3d5d16665734d06fbeeca58b7d98af","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-17 00:18:45.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/671195/20/check/DellEMC_Unity_FC/bf2e1f2/DellEMC_Unity_FC/1037 : FAILURE in 27m 01s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/671195/20/check/DellEMC_Unity_iSCSI/9d71844/DellEMC_Unity_iSCSI/989 : FAILURE in 28m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"85a90102e2d4e34ccfda50b5485e6b908e1fcf3d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-17 00:19:45.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/20/check/DellEMC_XtremIO_FC/5c62cd5/DellEMC_XtremIO_FC/1171 : FAILURE in 26m 43s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/20/check/DellEMC_XtremIO_iSCSI/c36aad2/DellEMC_XtremIO_iSCSI/1093 : FAILURE in 29m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":20},{"id":"8e5acaaf0cf4e2762b0215a846005557c9bb56b6","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-17 00:22:06.000000000","message":"Patch Set 20:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"589127feda8a69238a8a366bbd4b9f354bfd753c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-17 00:24:02.000000000","message":"Patch Set 20:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/671195/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"f6c1dcdd41ae1ef1fa64ce642555bfcae421bbdb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-17 00:35:34.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/20/iscsi/ : FAILURE in 44m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/20/smb/ : FAILURE in 44m 16s","accounts_in_message":[],"_revision_number":20},{"id":"3313fb8f1ea9f6a20ee3b6216faa1a60afd711bb","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-17 00:35:49.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/20/check/dsvm-tempest-kvm/1972160 : FAILURE in 34m 52s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/20/check/dsvm-tempest-exe/d6fe164 : FAILURE in 34m 33s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"466aa620330eab2d546bc8df684ee02ca193ce84","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-17 00:52:25.000000000","message":"Patch Set 18:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"52d35c137119102b05eb54466b785f0dd1714b69","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-17 00:57:22.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/20/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/26010a5 : FAILURE in 27s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/20/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/cf3dbb3 : SUCCESS in 1h 06m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":20},{"id":"7a95d69fa0a2689e4205494465501d13093b2b66","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-17 01:55:58.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/95/671195/20/check/NEC-ISCSI/8661143 : SUCCESS in 2h 03m 23s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/95/671195/20/check/NEC-FC/4bcf894 : SUCCESS in 2h 05m 08s","accounts_in_message":[],"_revision_number":20},{"id":"4493f88d14b5c2caa4ccc93063e5eaa7531a22f9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-17 02:13:27.000000000","message":"Patch Set 20:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/20/check/tempest-dsvm-full-xenial/944dbeb/ : SUCCESS in 2h 14m 54s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/20/check/devstack-plugin-ceph-dsvm-xenial/6f98385/ : SUCCESS in 2h 04m 52s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"aac372dd15be29d5d2ad7b30a4ba83f4ea862356","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 02:16:10.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32026b084a734fe9b5c32df6d7f1194c : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae64cd245744c4f930bdf585e478fb2 : SUCCESS in 8m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4301643ec7d4cb196d843c102f226a8 : SUCCESS in 10m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3be8899e2a414858abb5ebc9a286d2a7 : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72d494a3f578460ea777dfd7a4c18849 : FAILURE in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72913a28679c4ba58cdf545218a55f0b : SUCCESS in 9m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/815fe6e1082f4e28b2f139c3d34dc9c6 : SUCCESS in 1h 37m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/988df1c2eb054c888c2195f7f519c5df : SUCCESS in 55m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ed37e1b9fe234261ab81880314162a05 : SUCCESS in 1h 00m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/027432fe8a1e4743b0d610a30302497f : SUCCESS in 1h 17m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9ffa6caeefd4eb2adb368ab8280112e : SUCCESS in 17m 02s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d56224deecd940bea7fd66dc6564a61c : SUCCESS in 1h 33m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef1554241dd643d2ad0ebc5093fc0258 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7471a3af01844c3fa9bbf285a209d605 : SUCCESS in 5m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/181dc72f78644b2d9fbece6cafbdd856 : SUCCESS in 44m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a36fc45e09848b1b29d4f827835cfa2 : SUCCESS in 4m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a0e7a2c289234ad59122f646f3797765 : FAILURE in 1h 10m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2eb69c4f241a47a1bb0057b58d30b8b7 : SUCCESS in 5m 22s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7448493c449b44fa887c046edb3e8dfa : SUCCESS in 1h 04m 22s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d36e4102c64441518f875f074d721fe7 : SUCCESS in 1h 21m 55s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/771b216e44344db8adf62a0a48adbafd : SUCCESS in 1h 29m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f359eded35fb4452ad8d408af262f66e : SUCCESS in 2h 19m 25s","accounts_in_message":[],"_revision_number":20},{"id":"1b66390cda65125c6fa166d58e0c02f42bbc1ef3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-17 02:49:54.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/95/671195/20/check/DellEMC_PowerMAX_FC/25243cf/DellEMC_PowerMAX_FC/635 : FAILURE in 26m 37s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/95/671195/20/check/DellEMC_PowerMAX_iSCSI/9d3ac30/DellEMC_PowerMAX_iSCSI/611 : FAILURE in 21m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":20},{"id":"328fa04caeff2283ace265938b274d5037a1b09a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-17 04:06:02.000000000","message":"Patch Set 18:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/503/671195/18/check/infortrend-cinder-volume-iscsi/cae05bdb90146a1becc1737460d16de11e573a9f : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/484/671195/18/check/infortrend-cinder-volume-fc/cae05bdb90146a1becc1737460d16de11e573a9f : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"766e70ec5a9ee016565a4cc057266104ceb9cef6","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-17 06:46:54.000000000","message":"Patch Set 19:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/19 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"7f375f562248673670b08d7f9fbbee978a3be8a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-17 09:06:09.000000000","message":"Patch Set 18:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/504/671195/18/check/infortrend-cinder-volume-iscsi/cae05bdb90146a1becc1737460d16de11e573a9f : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/485/671195/18/check/infortrend-cinder-volume-fc/cae05bdb90146a1becc1737460d16de11e573a9f : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"c83a5fd84c3a681271b188ad75ee9d63ed53427c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-17 14:06:19.000000000","message":"Patch Set 19:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/505/671195/19/check/infortrend-cinder-volume-iscsi/534b5a686d1c113bd72a4a494dd6f9d281365d34 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/486/671195/19/check/infortrend-cinder-volume-fc/534b5a686d1c113bd72a4a494dd6f9d281365d34 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"7d224231fae3a865cfc34f5ac36e5140beee90c2","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-17 17:15:33.000000000","message":"Patch Set 20:\n\n(5 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"71a9dc3bf99a450801bc17a383a3391a569b6b69","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-17 17:42:08.000000000","message":"Patch Set 20:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"947231f34a3e0fa04a6c3d2d4304a3a9c4fcc665","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-17 19:06:21.000000000","message":"Patch Set 19:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/506/671195/19/check/infortrend-cinder-volume-iscsi/534b5a686d1c113bd72a4a494dd6f9d281365d34 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/487/671195/19/check/infortrend-cinder-volume-fc/534b5a686d1c113bd72a4a494dd6f9d281365d34 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"d913d8083beb3819b9d8fefaa22958309116e054","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-17 19:35:15.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"869c0a5d1d5ef24f71ca7129bfb29930b92d361f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-17 20:02:46.000000000","message":"Patch Set 20:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671195/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"c2bde97fdf5a7d5c127ebd121954e7937c697e32","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-17 20:20:35.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/20/iscsi/ : FAILURE in 44m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/20/smb/ : FAILURE in 44m 53s","accounts_in_message":[],"_revision_number":20},{"id":"c184f3f399e13a076a4c75bf797ce6ea4d2eb1e6","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-17 20:24:11.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/20/check/dsvm-tempest-kvm/2febccc : FAILURE in 34m 37s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/95/671195/20/check/dsvm-tempest-exe/732071d : FAILURE in 37m 06s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"7f699c1228e3929ec2fdcb5fc86a5085ff0901d5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-17 20:35:05.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/95/671195/20/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/068d9a6 : FAILURE in 29s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/95/671195/20/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/c735b49 : SUCCESS in 59m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":20},{"id":"f141ca97036e93956f5a0eb842915193fdaad32c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-17 21:18:27.000000000","message":"Patch Set 20:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/20/check/tempest-dsvm-full-xenial/4d0ffa6/ : SUCCESS in 1h 27m 38s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/671195/20/check/devstack-plugin-ceph-dsvm-xenial/d281a8c/ : SUCCESS in 1h 42m 53s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"287af0ec5e8db6395b45dc61c031cf51b04d6230","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 22:03:51.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e5ab537f20644c2a69da3f8f9ce1481 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4933626d88a455f9686a0eddf36fbe0 : SUCCESS in 8m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8537ddaef4794f6990867f06da9df610 : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66849a26c7394b38b4f962318717d947 : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3590d5793dc4c528bf9a6b33ebb8f1d : FAILURE in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4360b08402045edb07b301701b221ba : SUCCESS in 8m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a57ef4a6d11840f98840539078c6d145 : SUCCESS in 1h 47m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c09bff9111454fa68df685a884f9cf15 : FAILURE in 1h 00m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d4e4904f4440406ca178a9199770bb34 : SUCCESS in 1h 02m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e10fcdfe5d724ceca75bdc68a678fb26 : SUCCESS in 1h 27m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f8facbf7d2046709972deef5fe746c9 : SUCCESS in 18m 46s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c40a59bf637e4be19ca34e22e7374a3a : SUCCESS in 1h 25m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a6886a25776401ebed62211d4b30b49 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/369c64175baa4686a40eeb4ca5876cc8 : SUCCESS in 5m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/08be5de1873e4a8dadd3a5d53583294c : SUCCESS in 39m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c7999bbb7214624a5fdb6cb71953097 : SUCCESS in 5m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5540c214eb3340dfabaef4241ae6fb07 : SUCCESS in 1h 20m 21s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a124b995db8f481fa2fe76f542ca968c : SUCCESS in 5m 09s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/232524aa7c584f6b8e6db277306013f1 : SUCCESS in 1h 06m 15s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/062ce7e2ecef43b890a8712279f1da0c : SUCCESS in 1h 19m 37s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/9673188b9ad64ece94304f3256dfa168 : SUCCESS in 1h 19m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76b868169bc64201bae65b07ca632802 : SUCCESS in 2h 26m 12s","accounts_in_message":[],"_revision_number":20},{"id":"10234bb3bf162bcee740f3361d4192558c1fc569","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-17 22:09:13.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/671195/20/check/dsvm-tempest-cisco-zonemanager-job/8978149/ : SUCCESS in 1h 07m 19s","accounts_in_message":[],"_revision_number":20},{"id":"1e67ca656051b014b62fcd5f14b9638d5cbc7f78","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-18 00:06:26.000000000","message":"Patch Set 20:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/507/671195/20/check/infortrend-cinder-volume-iscsi/e18f05e735900acc12a7beac2ac8eca9a425e55a : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/488/671195/20/check/infortrend-cinder-volume-fc/e18f05e735900acc12a7beac2ac8eca9a425e55a : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"57fbc652fafa41462e1844407837c9f9cc2e349d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-18 05:07:21.000000000","message":"Patch Set 20:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/510/671195/20/check/infortrend-cinder-volume-iscsi/e18f05e735900acc12a7beac2ac8eca9a425e55a : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/491/671195/20/check/infortrend-cinder-volume-fc/e18f05e735900acc12a7beac2ac8eca9a425e55a : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"735b5ad9825580e741f475d4aa7af32a4838a7d5","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-19 13:29:31.000000000","message":"Patch Set 20:\n\nrecheck after kombu issue fixed (https://review.opendev.org/677070)","accounts_in_message":[],"_revision_number":20},{"id":"658535f651ffb8a164d96d8201eeadf3c8154ac6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-19 14:28:55.000000000","message":"Patch Set 20:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/511/671195/20/check/infortrend-cinder-volume-iscsi/e18f05e735900acc12a7beac2ac8eca9a425e55a : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/492/671195/20/check/infortrend-cinder-volume-fc/e18f05e735900acc12a7beac2ac8eca9a425e55a : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"850e59aa20391fb17a2d04d6ab5827b112108049","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-19 15:18:03.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/95/671195/20/check/dsvm-tempest-zadara-driver/51e5e15 : FAILURE in 1h 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ed29548c9e77d8a4628a6dfd56367c487e9d9440","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:30:20.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7ad3094bd8445e6aaa16ddea8d00a8b : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f45bea80089440f84a6bc5fea6ed28d : SUCCESS in 9m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd788257d82f4eea8c437fc74fa9632b : SUCCESS in 10m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7830df09569244378db74a6221ce8fea : SUCCESS in 11m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d2e2f01262849c497fe5c10d8d806f2 : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba8c87a50954e929eb6e58ea636f9f5 : SUCCESS in 8m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b4ce0375f487467a9f1e47279b28f877 : SUCCESS in 1h 37m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/32ad4808c89a4423b07718c346dc12a7 : SUCCESS in 50m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c078f1602c9744ffa585cac2afd4b38c : SUCCESS in 1h 08m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0f1eb26efea45afa41658660fdfe3bf : SUCCESS in 1h 49m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fc6b8ae8d0b470b8ed6994e66b68977 : SUCCESS in 16m 23s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5ad0178b48d446818f8dcdb8e226f3e7 : SUCCESS in 2h 00m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a0ce5ecfcde4777820bb900d446dd63 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/644432c15dd649269f28cf055d367a27 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4808240bdd9440d7b1036ea19768d67c : SUCCESS in 1h 04m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29758aab92904b2bb75373053e71ee92 : SUCCESS in 4m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47c082ef7c2440cc9affdd09795d96e7 : SUCCESS in 1h 26m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d84f9d09b109498c97f7fe7c6a1c560a : SUCCESS in 8m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fd817f5c121e44dbbef76fde600b13e8 : SUCCESS in 1h 41m 48s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9d9976b097e640ecaf148a4247748ac3 : FAILURE in 1h 16m 02s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/f230930bd0a74901909c52aa8ad76e4d : SUCCESS in 1h 31m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a09776d9bd4b46e9b21a3ab338f8d29a : SUCCESS in 2h 54m 24s","accounts_in_message":[],"_revision_number":20},{"id":"bed29c5c10961af51de5bf24540526e609e16fd8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-19 16:38:04.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/671195/20/check/dsvm-tempest-cisco-zonemanager-job/568a1ba/ : TIMED_OUT in 3h 00m 54s","accounts_in_message":[],"_revision_number":20},{"id":"3362199ff993ea557b2955e6ed58edc0adbe01c3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-19 17:33:49.000000000","message":"Patch Set 20:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-95-671195-20/ns5_iscsi : SUCCESS in 59m 13s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-95-671195-20/ns5_nfs : SUCCESS in 62m 47s","accounts_in_message":[],"_revision_number":20},{"id":"08bae3b42431989ed7240387c8221fdc1a37545e","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-19 17:35:06.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/95/671195/20/check/synology-dsm-cinder-iscsi/9dd75eb : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":20},{"id":"5c06fabec1d2ac205f748aa8a5b8ba0d6063a85d","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-19 17:41:58.000000000","message":"Patch Set 20:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/95/2545 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"494d3f28d5b5111b755de7ceec5fbb8e5e3ddda1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-19 17:54:41.000000000","message":"Patch Set 20:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-95-671195-20 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":20},{"id":"68d6891c67ece0963754313666a1841773fd4837","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-19 18:02:40.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/95/671195/20/upstream-check/cinder-cDOT-iSCSI/56bde45/ : SUCCESS in 1h 11m 12s\n- cinder-cDOT-NFS http://13.56.159.105/logs/95/671195/20/upstream-check/cinder-cDOT-NFS/0e1eb10/ : SUCCESS in 1h 09m 49s\n- cinder-cDOT-FCP http://13.56.159.105/logs/95/671195/20/upstream-check/cinder-cDOT-FCP/9d0a4c2/ : SUCCESS in 1h 31m 19s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"46fd737cf450faa60eef1063754c8547f39b199a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-19 18:06:20.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/95/671195/20/check/DellEMC_VNX_iSCSI/8e2fcab/DellEMC_VNX_iSCSI/600 : SUCCESS in 1h 35m 41s\n- DellEMC_VNX_FC http://publiclogs.emc.com/95/671195/20/check/DellEMC_VNX_FC/64c3b8f/DellEMC_VNX_FC/612 : FAILURE in 26m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":20},{"id":"a12658461b4d139ab904883dd6e2cece655453e2","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-19 18:07:55.000000000","message":"Patch Set 20:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/671195-20/HPMSA-FC : SUCCESS in 48:15.37\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/671195-20/HPMSA-iSCSI : SUCCESS in 51:06.13","accounts_in_message":[],"_revision_number":20},{"id":"b7ec02fecb31031619766b2f7d199f20f9c43e90","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-08-19 18:10:37.000000000","message":"Patch Set 20:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/671195-20/Lenovo-FC : SUCCESS in 47:23.86\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/671195-20/Lenovo-iSCSI : SUCCESS in 48:07.14","accounts_in_message":[],"_revision_number":20},{"id":"ab456481514425d03e531904aa1a98ee9e562be1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-19 18:24:29.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671195/20/iscsi/ : FAILURE in 4h 54m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/671195/20/smb/ : SUCCESS in 1h 36m 38s","accounts_in_message":[],"_revision_number":20},{"id":"aaecd1aa1249ab31f12336f520250d0c3de16ed9","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2019-08-19 18:27:22.000000000","message":"Patch Set 20:\n\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/671195-20/Seagate-FC : SUCCESS in 37:19.56\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/671195-20/Seagate-iSCSI : SUCCESS in 39:19.12","accounts_in_message":[],"_revision_number":20},{"id":"3b2cf6da6babb590b01bdc1429fd947513176da5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-19 18:28:11.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/95/671195/20/upstream-solidfire-check/cinder-solidfire-iSCSI/020754b/ : SUCCESS in 1h 56m 21s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":20},{"id":"63b53956c5e008ee89e64ec3e6b7bbdf65436cd8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-19 18:42:23.000000000","message":"Patch Set 20: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":20},{"id":"c5f839788addadffdd3e2506768a13c78016682a","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-08-19 18:45:53.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"48da8db4ce17cab856bb2602cc0d5fe8b5afce49","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-19 18:52:19.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"ae718afbd0b734ce29ee754c00253e9a3dc654f0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-19 19:07:56.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/671195/20/check/DellEMC_XtremIO_FC/c7903a7/DellEMC_XtremIO_FC/1175 : SUCCESS in 1h 29m 11s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/671195/20/check/DellEMC_XtremIO_iSCSI/aa2ba86/DellEMC_XtremIO_iSCSI/1097 : SUCCESS in 1h 29m 09s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":20},{"id":"4c67167048259c8d07589e1b9a1dc6aa202c7c97","author":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"date":"2019-08-19 20:01:11.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"9fdab45d8ba12036f6fac5ce3311de46046f50bb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-20 10:15:15.000000000","message":"Patch Set 20:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/20/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/34f030c : FAILURE in 46m 31s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/95/671195/20/check/3par-fc-driver-master-client-pip-ssa02-dsvm/0700af5 : FAILURE in 48m 03s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/95/671195/20/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/5618528 : FAILURE in 43m 26s","accounts_in_message":[],"_revision_number":20},{"id":"575845ec5b852458033a8272c338549464aaf35d","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-08-20 16:13:56.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"48c8989306c82cd415d3c08d669f171a8d235939","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:14:13.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"780706aea4fbec505b2e577ad001fc855fd7a6c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:06:01.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f85dbb22fae4a4a8cd843938244e66a : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b421efefacc54876a91ff5990045dc37 : SUCCESS in 9m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dddee70970d44a188359ea51a5291416 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ad5f5c9ccd464e87871f6c53e9db91 : SUCCESS in 12m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41091359d7af45f8b90ae6b2a9f5cc38 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9ded714b63421691863662ea9b58d9 : SUCCESS in 10m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4715eaa1aef24d9bb01215e1ea3bd224 : SUCCESS in 1h 42m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/088ab9b3e4f248eebbaf734873d1c886 : SUCCESS in 1h 12m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/be4506e15cbc4ff199204cdfe357c267 : SUCCESS in 1h 08m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fde1e23146ce4fc5bab9c8e833dd565e : SUCCESS in 1h 21m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a47e065e683545738e8f5d23263c3eef : SUCCESS in 19m 49s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a2638b20d77c4be0b3de48f0693fd8e6 : SUCCESS in 1h 19m 29s","accounts_in_message":[],"_revision_number":20},{"id":"7101d0ce698051a16e249d1c8971095e7c100058","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:06:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"e01c2fbb863d7daf2c327237919d702118f2919c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:08:07.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f7fe59a3c046cdaa79364493c90e5f : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"e18f05e735900acc12a7beac2ac8eca9a425e55a","revisions":{"cf5c205af2ebc6769eb586a9e3cb6d136e34dad7":{"kind":"REWORK","_number":1,"created":"2019-07-17 06:44:43.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/1"}}},"commit":{"parents":[{"commit":"3577a37edbb4760edc4a639693d2c3eade8f2093","subject":"Merge \"Update support matrix entries for MSA and Lenovo arrays.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3577a37edbb4760edc4a639693d2c3eade8f2093"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 06:38:24.000000000","tz":0},"subject":"WIP: Create Seagate driver from dothill driver","message":"WIP: Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf5c205af2ebc6769eb586a9e3cb6d136e34dad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf5c205af2ebc6769eb586a9e3cb6d136e34dad7"}]},"branch":"refs/heads/master"},"107fe0ea6f9865a0ef4b2a54e2302877ccc3645f":{"kind":"REWORK","_number":2,"created":"2019-07-18 05:13:10.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/2"}}},"commit":{"parents":[{"commit":"3577a37edbb4760edc4a639693d2c3eade8f2093","subject":"Merge \"Update support matrix entries for MSA and Lenovo arrays.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3577a37edbb4760edc4a639693d2c3eade8f2093"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-18 05:12:41.000000000","tz":0},"subject":"WIP: Create Seagate driver from dothill driver","message":"WIP: Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/107fe0ea6f9865a0ef4b2a54e2302877ccc3645f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/107fe0ea6f9865a0ef4b2a54e2302877ccc3645f"}]},"branch":"refs/heads/master"},"a845447c6ff0160b80dd0d09b16ab45f7dc8179d":{"kind":"REWORK","_number":3,"created":"2019-07-21 20:02:42.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/3"}}},"commit":{"parents":[{"commit":"feb174f35f021e91bac410df5ec2d3b5c7096500","subject":"Merge \"doc: Fix rbd driver marked support multiattach\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/feb174f35f021e91bac410df5ec2d3b5c7096500"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-21 19:45:50.000000000","tz":0},"subject":"WIP: Create Seagate driver from dothill driver","message":"WIP: Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\nDepends-on: https://review.opendev.org/671643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a845447c6ff0160b80dd0d09b16ab45f7dc8179d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a845447c6ff0160b80dd0d09b16ab45f7dc8179d"}]},"branch":"refs/heads/master"},"71c54edf1a658db6995976247fe8ee85c7a516fc":{"kind":"REWORK","_number":4,"created":"2019-07-22 02:38:18.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/4"}}},"commit":{"parents":[{"commit":"8ed3cfbe6437c10d6e0c4088224ea867c6223d7d","subject":"Fix dothill multiattach support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ed3cfbe6437c10d6e0c4088224ea867c6223d7d"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-22 02:27:53.000000000","tz":0},"subject":"WIP: Create Seagate driver from dothill driver","message":"WIP: Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\nDepends-on: https://review.opendev.org/671643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71c54edf1a658db6995976247fe8ee85c7a516fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71c54edf1a658db6995976247fe8ee85c7a516fc"}]},"branch":"refs/heads/master"},"940739c9d6d3f2e5e1ab1a9e27cf075cba845c44":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-22 05:44:02.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/5"}}},"commit":{"parents":[{"commit":"8ed3cfbe6437c10d6e0c4088224ea867c6223d7d","subject":"Fix dothill multiattach support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ed3cfbe6437c10d6e0c4088224ea867c6223d7d"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-22 05:38:25.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\nDepends-on: https://review.opendev.org/671643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/940739c9d6d3f2e5e1ab1a9e27cf075cba845c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/940739c9d6d3f2e5e1ab1a9e27cf075cba845c44"}]},"branch":"refs/heads/master"},"bc80ff14332eca2261efd2967b993dd324c9b769":{"kind":"REWORK","_number":6,"created":"2019-07-23 16:08:28.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/6"}}},"commit":{"parents":[{"commit":"fc628957d03365552dbf094ca143d3ae998ed154","subject":"Fix dothill multiattach support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc628957d03365552dbf094ca143d3ae998ed154"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-23 16:02:18.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\nDepends-on: https://review.opendev.org/671643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc80ff14332eca2261efd2967b993dd324c9b769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc80ff14332eca2261efd2967b993dd324c9b769"}]},"branch":"refs/heads/master"},"fadfd9a520531532e007503b9c4202ec39dcda46":{"kind":"REWORK","_number":7,"created":"2019-08-06 21:50:33.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/7"}}},"commit":{"parents":[{"commit":"b70c2237ef657c699225adfe804a9fd20b3e6601","subject":"Merge \"Update Veritas Access in Support Matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b70c2237ef657c699225adfe804a9fd20b3e6601"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-06 21:47:08.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fadfd9a520531532e007503b9c4202ec39dcda46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fadfd9a520531532e007503b9c4202ec39dcda46"}]},"branch":"refs/heads/master"},"1c7ba52932af249e2e8a61a9855805ad02833177":{"kind":"REWORK","_number":8,"created":"2019-08-07 01:25:23.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/8"}}},"commit":{"parents":[{"commit":"b70c2237ef657c699225adfe804a9fd20b3e6601","subject":"Merge \"Update Veritas Access in Support Matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b70c2237ef657c699225adfe804a9fd20b3e6601"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-07 01:24:29.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c7ba52932af249e2e8a61a9855805ad02833177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c7ba52932af249e2e8a61a9855805ad02833177"}]},"branch":"refs/heads/master"},"9e7caa2e65ae850dbfe3fe72e9a2e7c53854e808":{"kind":"REWORK","_number":9,"created":"2019-08-07 02:37:10.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/9"}}},"commit":{"parents":[{"commit":"b70c2237ef657c699225adfe804a9fd20b3e6601","subject":"Merge \"Update Veritas Access in Support Matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b70c2237ef657c699225adfe804a9fd20b3e6601"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-07 02:36:46.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e7caa2e65ae850dbfe3fe72e9a2e7c53854e808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e7caa2e65ae850dbfe3fe72e9a2e7c53854e808"}]},"branch":"refs/heads/master"},"fec61f5f09aa0c0d89bafc5407142501dcba25d3":{"kind":"REWORK","_number":10,"created":"2019-08-07 05:53:32.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/10"}}},"commit":{"parents":[{"commit":"0a0d55d8a95365f2a8b3bef861ab2410bec9e3ff","subject":"Fix upload volume to glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a0d55d8a95365f2a8b3bef861ab2410bec9e3ff"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-07 05:51:25.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fec61f5f09aa0c0d89bafc5407142501dcba25d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fec61f5f09aa0c0d89bafc5407142501dcba25d3"}]},"branch":"refs/heads/master"},"7df1c5d55b73aa72125d73a4338435aca1005f5c":{"kind":"REWORK","_number":11,"created":"2019-08-08 00:15:34.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/11"}}},"commit":{"parents":[{"commit":"0a0d55d8a95365f2a8b3bef861ab2410bec9e3ff","subject":"Fix upload volume to glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a0d55d8a95365f2a8b3bef861ab2410bec9e3ff"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-08 00:14:39.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7df1c5d55b73aa72125d73a4338435aca1005f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7df1c5d55b73aa72125d73a4338435aca1005f5c"}]},"branch":"refs/heads/master"},"85f327d6789ea4d8f4558122f26f45345bfbb422":{"kind":"REWORK","_number":12,"created":"2019-08-08 19:27:36.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/12"}}},"commit":{"parents":[{"commit":"0a0d55d8a95365f2a8b3bef861ab2410bec9e3ff","subject":"Fix upload volume to glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a0d55d8a95365f2a8b3bef861ab2410bec9e3ff"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-08 19:26:52.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85f327d6789ea4d8f4558122f26f45345bfbb422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85f327d6789ea4d8f4558122f26f45345bfbb422"}]},"branch":"refs/heads/master"},"6be5bcd2535d86011bbfc379c45116c926115c56":{"kind":"REWORK","_number":13,"created":"2019-08-08 21:02:30.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/13"}}},"commit":{"parents":[{"commit":"319760bb0ed8ec47027057304d88271dccbd6390","subject":"Merge \"Fix formatting issues in PowerMax docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/319760bb0ed8ec47027057304d88271dccbd6390"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-08 21:02:05.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6be5bcd2535d86011bbfc379c45116c926115c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6be5bcd2535d86011bbfc379c45116c926115c56"}]},"branch":"refs/heads/master"},"3f7c9889d78e594bcb3885e19b67a34e6aff1ffc":{"kind":"REWORK","_number":14,"created":"2019-08-09 03:06:04.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/14"}}},"commit":{"parents":[{"commit":"319760bb0ed8ec47027057304d88271dccbd6390","subject":"Merge \"Fix formatting issues in PowerMax docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/319760bb0ed8ec47027057304d88271dccbd6390"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-09 03:06:03.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f7c9889d78e594bcb3885e19b67a34e6aff1ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f7c9889d78e594bcb3885e19b67a34e6aff1ffc"}]},"branch":"refs/heads/master"},"1e456ea750ea4b28e073ecd35db15c0fec8de918":{"kind":"REWORK","_number":15,"created":"2019-08-13 17:46:45.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/15"}}},"commit":{"parents":[{"commit":"f426b7eb7a6c9e0acc40da044d56ea4d8961189c","subject":"Merge \"Fix kwargs passed to exception to get better format of error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f426b7eb7a6c9e0acc40da044d56ea4d8961189c"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-13 17:45:56.000000000","tz":-360},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e456ea750ea4b28e073ecd35db15c0fec8de918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e456ea750ea4b28e073ecd35db15c0fec8de918"}]},"branch":"refs/heads/master"},"4056dd658ade1bdfa67cf82b2e80d43ef0ce4b52":{"kind":"REWORK","_number":16,"created":"2019-08-16 03:07:29.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/16"}}},"commit":{"parents":[{"commit":"07ea0332f8da770271d0e19dff3ea813366ef04a","subject":"Merge \"Refactor API utilities into api_utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/07ea0332f8da770271d0e19dff3ea813366ef04a"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-16 02:58:57.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI and\ndeprecation of vendor-specific options requested by the core team.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected\nexcept for option-deprecation warnings.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4056dd658ade1bdfa67cf82b2e80d43ef0ce4b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4056dd658ade1bdfa67cf82b2e80d43ef0ce4b52"}]},"branch":"refs/heads/master"},"7c5d12440a35aac7c1fd398659c58b42dc9e8447":{"kind":"REWORK","_number":17,"created":"2019-08-16 06:11:58.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/17"}}},"commit":{"parents":[{"commit":"07ea0332f8da770271d0e19dff3ea813366ef04a","subject":"Merge \"Refactor API utilities into api_utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/07ea0332f8da770271d0e19dff3ea813366ef04a"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-16 06:11:21.000000000","tz":0},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI and\ndeprecation of vendor-specific options requested by the core team.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected\nexcept for option-deprecation warnings.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c5d12440a35aac7c1fd398659c58b42dc9e8447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c5d12440a35aac7c1fd398659c58b42dc9e8447"}]},"branch":"refs/heads/master"},"cae05bdb90146a1becc1737460d16de11e573a9f":{"kind":"REWORK","_number":18,"created":"2019-08-16 14:37:31.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/18"}}},"commit":{"parents":[{"commit":"64d63ca548516b0f33148788715b1501cff1db82","subject":"Merge \"Rollback the quota_usages table when failed to create a incremental backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64d63ca548516b0f33148788715b1501cff1db82"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-16 14:31:17.000000000","tz":-360},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI and\ndeprecation of vendor-specific options requested by the core team.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected\nexcept for option-deprecation warnings.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cae05bdb90146a1becc1737460d16de11e573a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cae05bdb90146a1becc1737460d16de11e573a9f"}]},"branch":"refs/heads/master"},"534b5a686d1c113bd72a4a494dd6f9d281365d34":{"kind":"REWORK","_number":19,"created":"2019-08-16 17:25:16.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/19"}}},"commit":{"parents":[{"commit":"64d63ca548516b0f33148788715b1501cff1db82","subject":"Merge \"Rollback the quota_usages table when failed to create a incremental backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64d63ca548516b0f33148788715b1501cff1db82"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-16 15:30:21.000000000","tz":-360},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI and\ndeprecation of vendor-specific options requested by the core team.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected\nexcept for option-deprecation warnings.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/534b5a686d1c113bd72a4a494dd6f9d281365d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/534b5a686d1c113bd72a4a494dd6f9d281365d34"}]},"branch":"refs/heads/master"},"e18f05e735900acc12a7beac2ac8eca9a425e55a":{"kind":"REWORK","_number":20,"created":"2019-08-16 23:50:24.000000000","uploader":{"_account_id":17042,"name":"Chris M","email":"cinder@tristero.net","username":"pots"},"ref":"refs/changes/95/671195/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/671195/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/671195/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/671195/20"}}},"commit":{"parents":[{"commit":"1962d17cade91744f0a92169bdec924c575e8613","subject":"Merge \"Mark Storpool Driver Unsupported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1962d17cade91744f0a92169bdec924c575e8613"}]}],"author":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-07-17 05:45:48.000000000","tz":0},"committer":{"name":"Chris M","email":"cinder@tristero.net","date":"2019-08-16 23:49:15.000000000","tz":-360},"subject":"Create Seagate driver from dothill driver","message":"Create Seagate driver from dothill driver\n\nThis patch creates a new \u0027supported\u0027 driver for Seagate (STX) FC and\niSCSI arrays by renaming and re-enabling the old \u0027unsupported\u0027 dothill\ndriver.  Other than marking the driver as \u0027supported\u0027, this patch\ncontains no changes in functionality except for multiattach fixes from\nchange I47f02729437cabab92ccc553a4c60d0c0a796952 needed to pass CI and\ndeprecation of vendor-specific options requested by the core team.\n\nOther drivers which referenced the dothill driver are modified to use\nthe Seagate class names, so users of those drivers will not be affected\nexcept for option-deprecation warnings.\n\nChange-Id: I3115ae296ae6b5702c7a8fa39249b8735542e17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e18f05e735900acc12a7beac2ac8eca9a425e55a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e18f05e735900acc12a7beac2ac8eca9a425e55a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
