)]}'
{"id":"openstack%2Fcinder~671235","triplet_id":"openstack%2Fcinder~master~Idfc8de789a27f54d9211be3c1ea6778586f85672","project":"openstack/cinder","branch":"master","topic":"ci_failure_fix","hashtags":[],"change_id":"Idfc8de789a27f54d9211be3c1ea6778586f85672","subject":"PowerMax Driver - SnapVX NoCopy Mode","status":"MERGED","created":"2019-07-17 10:17:23.000000000","updated":"2019-09-12 16:39:56.000000000","submitted":"2019-09-12 16:37:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671235-1568306275431-7eb2037b","meta_rev_id":"060a9cd3ad8ee4bb3ad40c6850dfa274922f8e4f","_number":671235,"virtual_id_number":671235,"owner":{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-31 02:53:05.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2019-08-31 15:32:51.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 02:27:34.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-30 22:14:43.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-30 15:11:58.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 06:05:32.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-12 16:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 14:31:36.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2019-08-31 03:27:04.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 21:40:29.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":30165,"name":"Simon O\u0027Donovan","email":"simon.odonovan@dell.com","username":"odonos12"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-08-31 22:34:31.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"date":"2019-08-31 14:26:25.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2019-08-31 08:44:49.000000000","_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-30 19:19:03.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2019-08-31 01:46:34.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 01:42:01.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 01:23:06.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-08-31 04:43:30.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"date":"2019-08-30 17:49:43.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-30 22:04:16.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2019-08-30 13:53:50.000000000","_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2019-08-31 03:34:11.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 23:10:32.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 21:19:51.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29627,"name":"MacroSAN Volume CI","email":"466367416@qq.com","username":"hjy"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"date":"2019-08-31 11:39:27.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2019-08-30 21:23:41.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-12 12:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":1,"date":"2019-09-04 14:25:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30165,"name":"Simon O\u0027Donovan","email":"simon.odonovan@dell.com","username":"odonos12"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-12 14:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29627,"name":"MacroSAN Volume CI","email":"466367416@qq.com","username":"hjy"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-02 09:39:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":30165,"name":"Simon O\u0027Donovan","email":"simon.odonovan@dell.com","username":"odonos12"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-12 14:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29627,"name":"MacroSAN Volume CI","email":"466367416@qq.com","username":"hjy"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":30165,"name":"Simon O\u0027Donovan","email":"simon.odonovan@dell.com","username":"odonos12"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29627,"name":"MacroSAN Volume CI","email":"466367416@qq.com","username":"hjy"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29627,"name":"MacroSAN Volume CI","email":"466367416@qq.com","username":"hjy"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":30165,"name":"Simon O\u0027Donovan","email":"simon.odonovan@dell.com","username":"odonos12"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 02:16:23.000000000","updated_by":{"_account_id":29627,"name":"MacroSAN Volume CI","email":"466367416@qq.com","username":"hjy"},"reviewer":{"_account_id":29627,"name":"MacroSAN Volume CI","email":"466367416@qq.com","username":"hjy"},"state":"REVIEWER"},{"updated":"2019-08-09 06:53:06.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2019-08-12 08:25:50.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-08-12 11:05:20.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-12 18:16:45.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2019-08-26 14:24:03.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-29 13:28:33.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-08-29 18:16:57.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-29 19:45:35.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-29 19:46:20.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 09:20:54.000000000","updated_by":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"reviewer":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"state":"REVIEWER"},{"updated":"2019-08-30 13:53:50.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 15:11:58.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 17:49:43.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-08-30 19:19:03.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-08-30 21:23:41.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 22:04:16.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-08-30 22:14:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 01:23:06.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-08-31 01:42:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 01:46:34.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 02:27:34.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 02:53:05.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 03:27:04.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 03:34:11.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 04:43:30.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 06:05:32.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 08:44:49.000000000","updated_by":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 11:39:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-08-31 14:26:25.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-08-31 14:31:36.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-08-31 15:32:51.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 21:19:51.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 21:40:29.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2019-08-31 22:34:31.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2019-08-31 23:10:32.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-02 09:39:49.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2019-09-04 14:25:42.000000000","updated_by":{"_account_id":30165,"name":"Simon O\u0027Donovan","email":"simon.odonovan@dell.com","username":"odonos12"},"reviewer":{"_account_id":30165,"name":"Simon O\u0027Donovan","email":"simon.odonovan@dell.com","username":"odonos12"},"state":"REVIEWER"},{"updated":"2019-09-12 12:51:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-12 14:40:36.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-09-12 16:37:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f9413b8d5e164d53fee08f8ac1bcb89795b18a8","author":{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},"date":"2019-07-17 10:17:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b5c6c96b0807b0cc8c4951b3c9c3d05137247a4","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-17 10:37:45.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1766 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1717 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1917 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ebf10cb6238f8315749f908827339f2c4896329c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-17 11:09:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/1/iscsi/ : SUCCESS in 51m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/1/smb/ : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":1},{"id":"796f08d2b0b7c47b8951380536bc518b5c0dc452","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-17 11:51:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/1/check/tempest-dsvm-full-xenial/a3ec311/ : SUCCESS in 1h 23m 00s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/1/check/devstack-plugin-ceph-dsvm-xenial/b6ee1ad/ : SUCCESS in 1h 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2733c98b8a5c295c8e96d984589faf9275f89e79","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-17 11:52:28.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/1/check/fujitsu-eternus-dx-iscsi/0c768b4 : SUCCESS in 1h 33m 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/1/check/fujitsu-eternus-dx-fc/c4362c1 : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":1},{"id":"076605bb4adc453c1b18fd323e69b9b9d2c3b81d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 11:57:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/1/check/dsvm-tempest-kvm/9e58f26 : FAILURE in 43m 37s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/1/check/dsvm-tempest-exe/cb4e9cf : FAILURE in 43m 20s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"d68e5eacec7e04dcd9fee5a5b6d89a2f63e59c33","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-17 11:59:03.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"dbaec2e5e59ec03a9897d660e4e8a8c592a8e878","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-17 12:25:50.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/1/check/DellEMC_Unity_FC/e0ad529/DellEMC_Unity_FC/694 : SUCCESS in 1h 06m 59s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/1/check/DellEMC_Unity_iSCSI/02ad830/DellEMC_Unity_iSCSI/660 : SUCCESS in 1h 24m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e9b038a2c05eda90740f903669042999f2619b2f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 13:23:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/1/check/DellEMC_XtremIO_FC/2fd925d/DellEMC_XtremIO_FC/865 : SUCCESS in 1h 28m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/1/check/DellEMC_XtremIO_iSCSI/54af236/DellEMC_XtremIO_iSCSI/812 : FAILURE in 1h 36m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f228aafda362b696e62835b3dfb1b3eaa5d7579","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:07:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/671235/1/check/openstack-tox-lower-constraints/9eff4a2/ : SUCCESS in 10m 43s\n- openstack-tox-pep8 http://logs.openstack.org/35/671235/1/check/openstack-tox-pep8/6e0b305/ : SUCCESS in 8m 06s\n- openstack-tox-py27 http://logs.openstack.org/35/671235/1/check/openstack-tox-py27/abdace6/ : SUCCESS in 7m 50s\n- openstack-tox-py36 http://logs.openstack.org/35/671235/1/check/openstack-tox-py36/7e23245/ : SUCCESS in 9m 15s\n- openstack-tox-py37 http://logs.openstack.org/35/671235/1/check/openstack-tox-py37/cc25325/ : SUCCESS in 9m 40s\n- openstack-tox-docs http://logs.openstack.org/35/671235/1/check/openstack-tox-docs/9522102/html/ : SUCCESS in 8m 23s\n- tempest-full http://logs.openstack.org/35/671235/1/check/tempest-full/5e30eb6/ : SUCCESS in 1h 35m 15s\n- neutron-grenade http://logs.openstack.org/35/671235/1/check/neutron-grenade/e2cd7dd/ : SUCCESS in 1h 17m 59s\n- grenade-py3 http://logs.openstack.org/35/671235/1/check/grenade-py3/cc2e3d7/ : SUCCESS in 1h 06m 29s\n- tempest-full-py3 http://logs.openstack.org/35/671235/1/check/tempest-full-py3/3b82a78/ : SUCCESS in 1h 49m 14s\n- build-openstack-releasenotes http://logs.openstack.org/35/671235/1/check/build-openstack-releasenotes/bb7c511/html/ : SUCCESS in 17m 48s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/35/671235/1/check/cinder-tempest-dsvm-lvm-lio-barbican/aca8cac/ : SUCCESS in 1h 27m 08s\n- cinder-tox-bandit-baseline http://logs.openstack.org/35/671235/1/check/cinder-tox-bandit-baseline/68da1f7/ : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/35/671235/1/check/openstack-tox-functional-py36/9314a9d/ : SUCCESS in 6m 12s\n- cinder-rally-task http://logs.openstack.org/35/671235/1/check/cinder-rally-task/ac8e76d/ : SUCCESS in 36m 10s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/35/671235/1/check/openstack-tox-pylint/b402414/ : SUCCESS in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/35/671235/1/check/cinder-plugin-ceph-tempest/3b5d1f0/ : SUCCESS in 1h 14m 50s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/35/671235/1/check/openstack-tox-functional/39064b1/ : SUCCESS in 4m 37s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/35/671235/1/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/22f990c/ : SUCCESS in 1h 18m 06s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/35/671235/1/check/legacy-tempest-dsvm-lvm-multibackend/519f31c/ : SUCCESS in 1h 17m 16s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/35/671235/1/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/66a0536/ : SUCCESS in 1h 26m 58s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/35/671235/1/check/tempest-slow-py3/5ef4f00/ : SUCCESS in 2h 09m 52s","accounts_in_message":[],"_revision_number":1},{"id":"299d7c562e47809e8de146abd2ab28c4b81bdcd3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-17 15:17:34.000000000","message":"Patch Set 1:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-35-671235-1/ns5_iscsi : FAILURE in 36m 16s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-35-671235-1/ns5_nfs : FAILURE in 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"700a2fa42cddc2159ed52a251812cf61f8f05c65","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-07-17 15:18:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/35/671235/1/upstream-solidfire-check/cinder-solidfire-iSCSI/947d53e/ : SUCCESS in 1h 09m 00s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"e969d22423d97f123fb2ce8432137ff62e8222ad","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-17 15:29:35.000000000","message":"Patch Set 1:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-671235-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"4d7a029c3ecf74d676f8a1b98591946923adc801","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-17 15:56:59.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/35/671235/1/check-not-vote/ext-cinder-zuul/3862c28 : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b67c31e3dd777c8561563da92016dcd6e4b59cae","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-07-17 16:19:02.000000000","message":"Patch Set 1:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/35/2365 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d19ca892ee696b15537024624a04dfdcc8f0b6b4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-17 16:21:31.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/fcfea34 : FAILURE in 2h 11m 18s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/1/check/3par-fc-driver-master-client-pip-ssa02-dsvm/c46e764 : FAILURE in 2h 00m 55s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/35/671235/1/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/1f7a204 : FAILURE in 1h 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5d70ba0f4e3196acf48a6d046db15e19617a9913","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-17 16:22:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.179.158/logs/35/671235/1/upstream-check/cinder-cDOT-iSCSI/6bfb8ff/ : SUCCESS in 2h 14m 28s\n- cinder-cDOT-NFS http://13.56.179.158/logs/35/671235/1/upstream-check/cinder-cDOT-NFS/2d4e0cd/ : FAILURE in 1h 25m 58s\n- cinder-cDOT-FCP http://13.56.179.158/logs/35/671235/1/upstream-check/cinder-cDOT-FCP/dc21e99/ : SUCCESS in 1h 23m 16s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"036442b440ac892907edd6a19725d4837a89bdaf","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-07-17 17:17:33.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/1/check/DellEMC_SC_FC/9f3cf9c/DellEMC_SC_FC/282 : SUCCESS in 1h 10m 47s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/1/check/DellEMC_SC_iSCSI/fff6a4e/DellEMC_SC_iSCSI/265 : SUCCESS in 1h 37m 30s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"44bbf46057b8f2c788e4bf0a86ee1d02be428c3b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 17:32:04.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/1/check/DellEMC_VNX_iSCSI/5aa6724/DellEMC_VNX_iSCSI/419 : SUCCESS in 1h 52m 21s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/1/check/DellEMC_VNX_FC/7166e61/DellEMC_VNX_FC/428 : SUCCESS in 1h 08m 13s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"28234d2d11bbc08afe0d8da031a26043dd4f5b86","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-07-17 17:37:07.000000000","message":"Patch Set 1:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/35/671235/1/HPMSA-FC : SUCCESS in 38:33.46\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/35/671235/1/HPMSA-iSCSI : SUCCESS in 35:37.44","accounts_in_message":[],"_revision_number":1},{"id":"db377f083b85f8ca3f8e9fe1372cb3c74b8f12a8","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-07-17 17:37:49.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/1/check/kaminario-dsvm-tempest-full-iscsi/52ecd5c : FAILURE in 33m 09s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/1/check/kaminario-dsvm-tempest-full-FC/8853465 : SUCCESS in 1h 48m 39s","accounts_in_message":[],"_revision_number":1},{"id":"97077a0f65f82b8a3a731951469fa501da2a31a3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-07-17 17:40:22.000000000","message":"Patch Set 1:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/35/671235/1/Lenovo-FC : SUCCESS in 38:12.05\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/35/671235/1/Lenovo-iSCSI : SUCCESS in 37:22.12","accounts_in_message":[],"_revision_number":1},{"id":"24b872cdfd726a7b286a1ee2f49ca430745fec76","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 17:50:07.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/35/671235/1/check/DellEMC_VxFlexOS/23f5538/DellEMC_VxFlexOS/443 : SUCCESS in 1h 51m 16s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"95eeee9fc4ce68b13924abb22c5f15672da4db27","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-17 18:08:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/671235/1/check/dsvm-tempest-cisco-zonemanager-job/4fc014c/ : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a815ccd405a0e5f9e4f2038470137a1aef71ef0c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-17 19:33:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/1/check/DellEMC_Unity_FC/4990900/DellEMC_Unity_FC/699 : FAILURE in 1h 10m 04s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/1/check/DellEMC_Unity_iSCSI/5fc94a5/DellEMC_Unity_iSCSI/665 : SUCCESS in 1h 23m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"eaff0dbc966f76d0249e2700dba3b7be998fd05a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 21:44:15.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/1/check/DellEMC_XtremIO_FC/f73ad59/DellEMC_XtremIO_FC/870 : SUCCESS in 1h 27m 04s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/1/check/DellEMC_XtremIO_iSCSI/f2b44f4/DellEMC_XtremIO_iSCSI/817 : SUCCESS in 1h 32m 32s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"5a0a02410f7e0262d2badf2970e862b495aa2bbf","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-07-17 21:47:08.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/35/671235/1/check/Cinder-AS13000/033dcc4 : SUCCESS in 2h 22m 18s\n- cinder-test-fc http://218.29.111.78:9000/35/671235/1/check/cinder-test-fc/7c06f74 : FAILURE in 37m 46s\n- cinder-test-iscsi http://218.29.111.78:9000/35/671235/1/check/cinder-test-iscsi/c311303 : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bd41839012b4aaad9446ddb8e13673e27b10e1a1","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 23:34:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/1/check/DellEMC_PowerMAX_FC/0a65abe/DellEMC_PowerMAX_FC/434 : FAILURE in 2h 31m 52s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/1/check/DellEMC_PowerMAX_iSCSI/5e578fe/DellEMC_PowerMAX_iSCSI/419 : FAILURE in 2h 21m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"a1f31616847094a69b9edf38635f97276048225b","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-07-18 10:16:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"98e26291d6d54fff90925d8f9257b89c8c658994","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-07-18 10:26:31.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1731 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"815b36b02468819d2b9f858707c838340467c738","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-18 11:47:42.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/2/check/tempest-dsvm-full-xenial/13c3396/ : SUCCESS in 1h 27m 15s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/2/check/devstack-plugin-ceph-dsvm-xenial/dd62144/ : SUCCESS in 1h 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36f0208a05714259bc973c6723c28025e5c52b33","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 13:27:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/2/check/dsvm-tempest-kvm/e9dab7f : FAILURE in 40m 59s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/2/check/dsvm-tempest-exe/3423e46 : FAILURE in 42m 01s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"b67505a03e74dff951c3355009dd428a22ed2e74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 13:40:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/671235/2/check/openstack-tox-lower-constraints/aad2021/ : SUCCESS in 11m 45s\n- openstack-tox-pep8 http://logs.openstack.org/35/671235/2/check/openstack-tox-pep8/009a82f/ : SUCCESS in 7m 50s\n- openstack-tox-py27 http://logs.openstack.org/35/671235/2/check/openstack-tox-py27/3243c42/ : SUCCESS in 8m 36s\n- openstack-tox-py36 http://logs.openstack.org/35/671235/2/check/openstack-tox-py36/a61f77c/ : SUCCESS in 8m 20s\n- openstack-tox-py37 http://logs.openstack.org/35/671235/2/check/openstack-tox-py37/2cccb9e/ : SUCCESS in 8m 29s\n- openstack-tox-docs http://logs.openstack.org/35/671235/2/check/openstack-tox-docs/362c7f7/html/ : SUCCESS in 7m 56s\n- tempest-full http://logs.openstack.org/35/671235/2/check/tempest-full/e156889/ : SUCCESS in 1h 32m 32s\n- neutron-grenade http://logs.openstack.org/35/671235/2/check/neutron-grenade/32c24f0/ : SUCCESS in 59m 47s\n- grenade-py3 http://logs.openstack.org/35/671235/2/check/grenade-py3/5a15421/ : SUCCESS in 1h 01m 32s\n- tempest-full-py3 http://logs.openstack.org/35/671235/2/check/tempest-full-py3/3d16abe/ : SUCCESS in 1h 38m 36s\n- build-openstack-releasenotes http://logs.openstack.org/35/671235/2/check/build-openstack-releasenotes/0f1536d/html/ : SUCCESS in 19m 46s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/35/671235/2/check/cinder-tempest-dsvm-lvm-lio-barbican/45f25a3/ : SUCCESS in 1h 42m 06s\n- cinder-tox-bandit-baseline http://logs.openstack.org/35/671235/2/check/cinder-tox-bandit-baseline/80ef908/ : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/35/671235/2/check/openstack-tox-functional-py36/d48e336/ : SUCCESS in 6m 25s\n- cinder-rally-task http://logs.openstack.org/35/671235/2/check/cinder-rally-task/6c74107/ : SUCCESS in 42m 28s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/35/671235/2/check/openstack-tox-pylint/5b54168/ : SUCCESS in 3m 55s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/35/671235/2/check/cinder-plugin-ceph-tempest/ea9cdd2/ : SUCCESS in 1h 09m 11s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/35/671235/2/check/openstack-tox-functional/76b37cd/ : SUCCESS in 5m 14s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/35/671235/2/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/162d3f4/ : SUCCESS in 1h 04m 25s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/35/671235/2/check/legacy-tempest-dsvm-lvm-multibackend/2090492/ : SUCCESS in 1h 35m 39s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/35/671235/2/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/d15bad0/ : SUCCESS in 1h 22m 48s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/35/671235/2/check/tempest-slow-py3/cc7451a/ : SUCCESS in 2h 38m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f91b9fd358568ac2795066d7aa6112080a0e0150","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-18 13:46:03.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"334027ca851ed3136a52570d6058c4487e5ba4fc","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-18 13:56:00.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/2/check/fujitsu-eternus-dx-iscsi/b75f043 : SUCCESS in 1h 35m 05s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/2/check/fujitsu-eternus-dx-fc/5549f68 : SUCCESS in 1h 59m 21s","accounts_in_message":[],"_revision_number":2},{"id":"44f8e352ea4e3456bf80ad8eab6eda987d9f3958","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-18 14:22:39.000000000","message":"Patch Set 2:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-35-671235-2/ns5_iscsi : FAILURE in 37m 59s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-35-671235-2/ns5_nfs : FAILURE in 37m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9e390599780834d3c4db5de63d43e8266b4953c5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-18 14:25:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/2/iscsi/ : SUCCESS in 1h 04m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/2/smb/ : FAILURE in 59m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d24cc32802d616ab17c6f6ddd0c4f3d3822b5b95","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-18 14:38:00.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/db3eb7e : FAILURE in 53m 52s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/2/check/3par-fc-driver-master-client-pip-ssa02-dsvm/c4581ea : FAILURE in 46m 29s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/35/671235/2/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/c548a75 : FAILURE in 38m 29s","accounts_in_message":[],"_revision_number":2},{"id":"95a656a2be529af063e540d002f141d20a045356","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-18 15:00:05.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-671235-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"cac0a49fba0f703a88d4169b5d9c42d75ef83e40","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-07-18 15:21:46.000000000","message":"Patch Set 2:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/35/2370 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"de4c0f53111078dd85231161b4a03e7e99ce6c99","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 15:32:56.000000000","message":"Patch Set 2:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-35-671235-2-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"54d6908042b3c550339ba1783caeaf9aed995f79","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-07-18 16:03:07.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/2/check/kaminario-dsvm-tempest-full-iscsi/07e9d8a : SUCCESS in 2h 19m 51s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/2/check/kaminario-dsvm-tempest-full-FC/eca6036 : FAILURE in 1m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f22778cd904ab105e13f4e2f0df48b8b2c9be125","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-18 16:22:06.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/2/check/DellEMC_Unity_FC/2051f2f/DellEMC_Unity_FC/713 : SUCCESS in 1h 07m 23s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/2/check/DellEMC_Unity_iSCSI/bb8bbe2/DellEMC_Unity_iSCSI/679 : SUCCESS in 1h 33m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"774d7609b3eca32e203b3f38a041ff7c92d3b95b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-07-18 16:25:26.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/2/check/DellEMC_SC_FC/9b3acb8/DellEMC_SC_FC/287 : SUCCESS in 1h 09m 50s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/2/check/DellEMC_SC_iSCSI/5bc30ad/DellEMC_SC_iSCSI/270 : SUCCESS in 1h 47m 15s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"e6d66bffcee219fe1785d9595e7a2b3abff1394c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 16:42:33.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/35/671235/2/check/DellEMC_VxFlexOS/8bb2e8e/DellEMC_VxFlexOS/448 : SUCCESS in 1h 56m 39s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"0ee4deb369476305903b3b8f236512e37d4fb211","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-18 16:55:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://52.53.230.100/logs/35/671235/2/upstream-check/cinder-cDOT-iSCSI/9d56d31/ : SUCCESS in 1h 02m 46s\n- cinder-cDOT-NFS http://52.53.230.100/logs/35/671235/2/upstream-check/cinder-cDOT-NFS/4856407/ : FAILURE in 3h 12m 38s\n- cinder-cDOT-FCP http://52.53.230.100/logs/35/671235/2/upstream-check/cinder-cDOT-FCP/b551b74/ : FAILURE in 1h 57m 40s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"bae7cf600ff34302d381359cf6174e69d207aa59","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-07-18 17:08:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://52.53.230.100/logs/35/671235/2/upstream-solidfire-check/cinder-solidfire-iSCSI/158812c/ : SUCCESS in 3h 24m 49s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"90fd2ecbdde30d931f21fda4eccd80e0f3da3b54","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-07-18 17:21:24.000000000","message":"Patch Set 2:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/35/671235/2/Lenovo-FC : SUCCESS in 37:30.24\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/35/671235/2/Lenovo-iSCSI : SUCCESS in 36:23.52","accounts_in_message":[],"_revision_number":2},{"id":"bc96f96b9820882a1579700ddd73c1625e812ed0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 17:22:38.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/2/check/DellEMC_VNX_iSCSI/6d5edbe/DellEMC_VNX_iSCSI/424 : FAILURE in 2h 50m 12s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/2/check/DellEMC_VNX_FC/8af2cd0/DellEMC_VNX_FC/433 : SUCCESS in 1h 03m 50s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"c12ff9b98dd573cf294480a541b943145d3fed5e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-07-18 17:35:47.000000000","message":"Patch Set 2:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/35/671235/2/HPMSA-FC : SUCCESS in 45:17.08\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/35/671235/2/HPMSA-iSCSI : SUCCESS in 41:24.31","accounts_in_message":[],"_revision_number":2},{"id":"5a76ec120ebe9963b3246707235d3206ecd39e50","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-18 18:07:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/671235/2/check/dsvm-tempest-cisco-zonemanager-job/8e8f499/ : SUCCESS in 1h 26m 00s","accounts_in_message":[],"_revision_number":2},{"id":"47c48419e39c554138223d4dae8c64ad5efa7920","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-07-18 18:21:09.000000000","message":"Patch Set 2: Code-Review+1\n\nCode looks ok but don\u0027t see a powermax CI result.","accounts_in_message":[],"_revision_number":2},{"id":"b35996369af66b0dfdd509f9ecfb0e90d3cf684a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-19 00:00:01.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/35/671235/2/check-cinder/Cinder-ISER-ISCSI/3c7f3e5 : FAILURE in 1h 08m 02s\n- Cinder-ISER-LIO http://13.74.249.42/35/671235/2/check-cinder/Cinder-ISER-LIO/89d599f : SUCCESS in 1h 07m 38s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"e2ceab05ecc9d3e56fdc5f7a84a468d942deaa5d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-19 00:42:39.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/35/671235/2/check-not-vote/ext-cinder-zuul/1ab0a00 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":2},{"id":"1660ee311f3c528fa13d34422e42c460734feac0","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-07-23 02:16:33.000000000","message":"Patch Set 2:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":2},{"id":"cce6ebc048fe02268d35727e0259f32164160223","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-23 12:41:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/2/check/DellEMC_PowerMAX_FC/2c7ef6e/DellEMC_PowerMAX_FC/476 : SUCCESS in 3h 45m 21s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/2/check/DellEMC_PowerMAX_iSCSI/16e81ed/DellEMC_PowerMAX_iSCSI/457 : SUCCESS in 5h 15m 56s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"94e3d22b8f19c9cf206999cc27220fe97d9f452b","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-07-24 09:19:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5812638d78105c0f2703280eaee3e3a4aca3689a","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-07-30 20:43:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9a9e4611d112a931621dd778f6a4ba28350de020","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-30 22:25:41.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/3/iscsi/ : SUCCESS in 55m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/3/smb/ : FAILURE in 53m 55s","accounts_in_message":[],"_revision_number":3},{"id":"50fe67f8704d4a3ac6297559a4ded6334599798a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-30 22:28:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/3/check/fujitsu-eternus-dx-iscsi/5e07312 : FAILURE in 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/3/check/fujitsu-eternus-dx-fc/7bf3eed : FAILURE in 51s","accounts_in_message":[],"_revision_number":3},{"id":"5c6470d7db7bfdf6b1b35a942af05d7c9faf36f0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-30 22:34:51.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/3/check/tempest-dsvm-full-xenial/21d6cd8/ : SUCCESS in 1h 32m 34s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/3/check/devstack-plugin-ceph-dsvm-xenial/6da54ec/ : SUCCESS in 1h 50m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"692b8fd55a63a1aaeaf6c4e1e7dd1667dd46bdda","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-07-30 22:36:55.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1861 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b4329cd293264218c6d74f6327435962c33d9bf4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-30 23:03:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c50cd95 : SUCCESS in 57m 44s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e4f951b : SUCCESS in 1h 12m 33s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"ce99b619eebfadd25444746b542e56a9f51c3e9a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-30 23:59:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/3/check/dsvm-tempest-kvm/13b4742 : FAILURE in 34m 44s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/3/check/dsvm-tempest-exe/538c3d3 : FAILURE in 33m 56s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"06fe6ad8dece956a396dc3e21eea6cbb0991289d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 00:56:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/671235/3/check/openstack-tox-lower-constraints/a007a50/ : SUCCESS in 12m 13s\n- openstack-tox-pep8 http://logs.openstack.org/35/671235/3/check/openstack-tox-pep8/5de381d/ : SUCCESS in 8m 23s\n- openstack-tox-py27 http://logs.openstack.org/35/671235/3/check/openstack-tox-py27/db75f6a/ : SUCCESS in 9m 25s\n- openstack-tox-py36 http://logs.openstack.org/35/671235/3/check/openstack-tox-py36/46af7f2/ : SUCCESS in 8m 56s\n- openstack-tox-py37 http://logs.openstack.org/35/671235/3/check/openstack-tox-py37/e656ca1/ : SUCCESS in 9m 08s\n- openstack-tox-docs http://logs.openstack.org/35/671235/3/check/openstack-tox-docs/a779aff/html/ : SUCCESS in 9m 51s\n- tempest-full http://logs.openstack.org/35/671235/3/check/tempest-full/e013746/ : SUCCESS in 1h 35m 33s\n- neutron-grenade http://logs.openstack.org/35/671235/3/check/neutron-grenade/04bfe6c/ : SUCCESS in 57m 54s\n- grenade-py3 http://logs.openstack.org/35/671235/3/check/grenade-py3/a06e535/ : SUCCESS in 1h 05m 20s\n- tempest-full-py3 http://logs.openstack.org/35/671235/3/check/tempest-full-py3/23e8ee1/ : SUCCESS in 1h 21m 23s\n- build-openstack-releasenotes http://logs.openstack.org/35/671235/3/check/build-openstack-releasenotes/97da64c/html/ : SUCCESS in 15m 14s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/35/671235/3/check/cinder-tempest-dsvm-lvm-lio-barbican/21ac5fc/ : SUCCESS in 1h 33m 28s\n- cinder-tox-bandit-baseline http://logs.openstack.org/35/671235/3/check/cinder-tox-bandit-baseline/6a01ea0/ : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/35/671235/3/check/openstack-tox-functional-py36/7c2c0b4/ : SUCCESS in 4m 49s\n- cinder-rally-task http://logs.openstack.org/35/671235/3/check/cinder-rally-task/7e15d74/ : SUCCESS in 40m 15s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/35/671235/3/check/openstack-tox-pylint/46ffac7/ : SUCCESS in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/35/671235/3/check/cinder-plugin-ceph-tempest/72507ec/ : SUCCESS in 1h 29m 08s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/35/671235/3/check/openstack-tox-functional/2606503/ : SUCCESS in 5m 38s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/35/671235/3/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/0df184a/ : SUCCESS in 1h 03m 24s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/35/671235/3/check/legacy-tempest-dsvm-lvm-multibackend/e0174bc/ : SUCCESS in 1h 26m 32s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/35/671235/3/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/d8a4631/ : SUCCESS in 1h 40m 09s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/35/671235/3/check/tempest-slow-py3/e8342ea/ : SUCCESS in 2h 23m 04s","accounts_in_message":[],"_revision_number":3},{"id":"2be54179bda8f14ec57ac43e0cb96c90f3abf6a8","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-07-31 01:16:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/35/671235/3/upstream-solidfire-check/cinder-solidfire-iSCSI/80a0191/ : FAILURE in 13m 15s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"ff09bc17c601cb83c935f07fb99d27a0b20e2902","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-31 01:28:48.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"db46b48b851948f13c7fee78037a8536d1981ee5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-31 05:19:46.000000000","message":"Patch Set 3:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/daed69b : SUCCESS in 2h 03m 06s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/3/check/3par-fc-driver-master-client-pip-ssa02-dsvm/eab1329 : SUCCESS in 2h 20m 01s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/35/671235/3/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/79ccb46 : FAILURE in 2h 01m 29s","accounts_in_message":[],"_revision_number":3},{"id":"40d97630dfbe7a224941ef6b250be77da77cfa0f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-31 05:34:31.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/3/check/DellEMC_Unity_FC/9a35c58/DellEMC_Unity_FC/863 : FAILURE in 1h 13m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/3/check/DellEMC_Unity_iSCSI/361b53c/DellEMC_Unity_iSCSI/824 : SUCCESS in 1h 37m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ad3df2badeece48e72ec55c7b5e54aa8e53a7db6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-31 05:38:18.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-671235-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"1975b920cf7b3003e1ab0a6c03b81de60c44911f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-07-31 05:41:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/35/671235/3/check/dsvm-tempest-zadara-driver/e3e1ac4 : SUCCESS in 1h 41m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0a16ba4a0fb2cc73106e4867896c387f6250dcfe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-31 05:48:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/671235/3/check/dsvm-tempest-cisco-zonemanager-job/f615a76/ : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":3},{"id":"3e4ebd7b3c469ff2cc42f5cdb7aa2ad272ebcfe6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-31 05:48:58.000000000","message":"Patch Set 3:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-35-671235-3/ns5_iscsi : FAILURE in 50m 31s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-35-671235-3/ns5_nfs : FAILURE in 53m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3353e830eb50e13a384856ef41fb62a0720eacf2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-31 07:17:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.161.150/logs/35/671235/3/upstream-check/cinder-cDOT-iSCSI/93b46fb/ : SUCCESS in 56m 59s\n- cinder-cDOT-NFS http://13.56.161.150/logs/35/671235/3/upstream-check/cinder-cDOT-NFS/083d7ff/ : SUCCESS in 59m 27s\n- cinder-cDOT-FCP http://13.56.161.150/logs/35/671235/3/upstream-check/cinder-cDOT-FCP/98b138c/ : SUCCESS in 1h 04m 58s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"dafc2e0d02f768b595c80c7d49f9bfbbb8ac6ab4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 07:40:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/3/check/DellEMC_XtremIO_FC/7126282/DellEMC_XtremIO_FC/994 : SUCCESS in 1h 26m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/3/check/DellEMC_XtremIO_iSCSI/9f796cb/DellEMC_XtremIO_iSCSI/928 : SUCCESS in 1h 52m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"56ba3b16e3f2e2761f0e9f54a3ca4e8758fabbf0","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 08:25:40.000000000","message":"Patch Set 3:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-35-671235-3-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"d9ed4f88ae277f664fbfd14510f7caf3295818a0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-31 08:43:24.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/35/671235/3/check-not-vote/ext-cinder-zuul/b191ab8 : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":3},{"id":"5cefeb387b9ede9725be8080fb980891da782e31","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-07-31 08:59:14.000000000","message":"Patch Set 3:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/35/2444 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"861a788362fdded4b0f329094d3fb5a78b1970a7","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-07-31 12:19:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/3/check/kaminario-dsvm-tempest-full-iscsi/65e3a27 : SUCCESS in 1h 56m 03s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/3/check/kaminario-dsvm-tempest-full-FC/507966e : SUCCESS in 1h 59m 15s","accounts_in_message":[],"_revision_number":3},{"id":"0c78e27433b27c8580b4d49f17887c5d4ff4d704","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 12:39:38.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/3/check/DellEMC_VNX_iSCSI/8d80254/DellEMC_VNX_iSCSI/495 : SUCCESS in 2h 00m 18s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/3/check/DellEMC_VNX_FC/a045323/DellEMC_VNX_FC/505 : SUCCESS in 1h 04m 42s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"382f43805b8bd701ba25da4e6c7bb4749ce8e5a6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-07-31 17:39:57.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/3/check/DellEMC_SC_FC/74021b6/DellEMC_SC_FC/359 : SUCCESS in 1h 10m 10s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/3/check/DellEMC_SC_iSCSI/035691e/DellEMC_SC_iSCSI/342 : FAILURE in 2h 52m 31s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"5fe6429828dac45a885893fc9e531acf2340c5e4","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-07-31 17:55:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"167ee12c8b5d85b8b62d9716b0ab34523d0c99b4","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-07-31 18:38:22.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1899 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1876 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/2050 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"4c932d0f01b5a9e1ae3a40ef2fd72617cdc9f27b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-31 18:53:01.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/4/iscsi/ : FAILURE in 56m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/4/smb/ : FAILURE in 53m 28s","accounts_in_message":[],"_revision_number":4},{"id":"c26a68b9af95a78e76ecf20fbfdaf072ecc68e12","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 19:16:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/4/check/dsvm-tempest-kvm/f16ad74 : FAILURE in 34m 25s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/4/check/dsvm-tempest-exe/adb30f9 : FAILURE in 34m 03s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"172a3abbe39018feb83217ce00f4ea4cb49fa7ee","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-31 19:19:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/4/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f3cb107 : SUCCESS in 1h 11m 15s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/4/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/523f926 : SUCCESS in 1h 23m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"baa3b19c2dc97fd93f1029151da04814927df304","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-31 19:29:05.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"8da18302ed520e2a501d80cd731f87136e4b5a98","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-07-31 19:43:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/4/check/fujitsu-eternus-dx-iscsi/1b85537 : SUCCESS in 1h 47m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/4/check/fujitsu-eternus-dx-fc/ba17bea : FAILURE in 53s","accounts_in_message":[],"_revision_number":4},{"id":"db58ae8aff2e19c7eb5de530c1ed475b97a2fec8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-31 19:57:39.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/4/check/tempest-dsvm-full-xenial/f004954/ : SUCCESS in 1h 56m 50s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/4/check/devstack-plugin-ceph-dsvm-xenial/b2226f7/ : SUCCESS in 2h 00m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9a98a4558e940df19a764506fd93fd8f84425f3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 22:11:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/671235/4/check/openstack-tox-lower-constraints/158692f/ : SUCCESS in 15m 17s\n- openstack-tox-pep8 http://logs.openstack.org/35/671235/4/check/openstack-tox-pep8/87ecd80/ : SUCCESS in 11m 46s\n- openstack-tox-py27 http://logs.openstack.org/35/671235/4/check/openstack-tox-py27/73f2f1d/ : SUCCESS in 9m 40s\n- openstack-tox-py36 http://logs.openstack.org/35/671235/4/check/openstack-tox-py36/e0a2378/ : SUCCESS in 12m 10s\n- openstack-tox-py37 http://logs.openstack.org/35/671235/4/check/openstack-tox-py37/e338932/ : SUCCESS in 11m 03s\n- openstack-tox-docs http://logs.openstack.org/35/671235/4/check/openstack-tox-docs/3f78545/html/ : SUCCESS in 8m 37s\n- tempest-full http://logs.openstack.org/35/671235/4/check/tempest-full/e76ab86/ : SUCCESS in 1h 48m 16s\n- neutron-grenade http://logs.openstack.org/35/671235/4/check/neutron-grenade/e9c954b/ : SUCCESS in 1h 03m 14s\n- grenade-py3 http://logs.openstack.org/35/671235/4/check/grenade-py3/f601415/ : SUCCESS in 1h 05m 20s\n- tempest-full-py3 http://logs.openstack.org/35/671235/4/check/tempest-full-py3/7206dfd/ : SUCCESS in 1h 29m 02s\n- build-openstack-releasenotes http://logs.openstack.org/35/671235/4/check/build-openstack-releasenotes/8a1dcf9/html/ : SUCCESS in 16m 10s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/35/671235/4/check/cinder-tempest-dsvm-lvm-lio-barbican/32919e4/ : SUCCESS in 1h 32m 31s\n- cinder-tox-bandit-baseline http://logs.openstack.org/35/671235/4/check/cinder-tox-bandit-baseline/e8e34cb/ : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/35/671235/4/check/openstack-tox-functional-py36/06499fe/ : SUCCESS in 6m 29s\n- cinder-rally-task http://logs.openstack.org/35/671235/4/check/cinder-rally-task/36c832b/ : SUCCESS in 42m 38s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/35/671235/4/check/openstack-tox-pylint/aead18d/ : SUCCESS in 4m 44s (non-voting)\n- cinder-plugin-ceph-tempest http://logs.openstack.org/35/671235/4/check/cinder-plugin-ceph-tempest/9ae0d04/ : SUCCESS in 1h 14m 08s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/35/671235/4/check/openstack-tox-functional/663623a/ : SUCCESS in 6m 02s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/35/671235/4/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/c20fd02/ : SUCCESS in 1h 10m 45s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/35/671235/4/check/legacy-tempest-dsvm-lvm-multibackend/b3e0f09/ : SUCCESS in 1h 36m 56s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/35/671235/4/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/2b61f29/ : SUCCESS in 1h 33m 42s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/35/671235/4/check/tempest-slow-py3/e9bd566/ : SUCCESS in 2h 31m 47s","accounts_in_message":[],"_revision_number":4},{"id":"ad2ee7725d68806d6e9dd7c425db719eda665ab8","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-07-31 23:42:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/35/671235/4/upstream-solidfire-check/cinder-solidfire-iSCSI/2eb567c/ : SUCCESS in 1h 21m 57s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"53f9cc7e1cdce5005f32ee818a2115a0c388f9fa","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-31 23:46:21.000000000","message":"Patch Set 4:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-35-671235-4/ns5_iscsi : SUCCESS in 48m 33s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-35-671235-4/ns5_nfs : FAILURE in 55m 44s","accounts_in_message":[],"_revision_number":4},{"id":"d35aa48e9c8e51f50d15c7abd098412afbc0fa3e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-31 23:47:21.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-671235-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"3f33e4e455c03a53c941f50361ad4ab0eb95a9c8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-01 00:01:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.161.150/logs/35/671235/4/upstream-check/cinder-cDOT-iSCSI/cec192c/ : SUCCESS in 1h 03m 38s\n- cinder-cDOT-NFS http://13.56.161.150/logs/35/671235/4/upstream-check/cinder-cDOT-NFS/10a1c47/ : SUCCESS in 1h 06m 07s\n- cinder-cDOT-FCP http://13.56.161.150/logs/35/671235/4/upstream-check/cinder-cDOT-FCP/b1fee9e/ : FAILURE in 1h 42m 52s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"159f934b8f80a90fdab80427a796c0d0f659c0bc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-01 00:22:47.000000000","message":"Patch Set 4:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e700003 : SUCCESS in 2h 03m 47s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/4/check/3par-fc-driver-master-client-pip-ssa02-dsvm/7e13212 : SUCCESS in 2h 08m 31s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/35/671235/4/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/79076a9 : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":4},{"id":"3d5acce7d7963a0cd1af1b3b0df26b9996e33184","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-01 00:24:05.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/4/check/DellEMC_Unity_FC/452801e/DellEMC_Unity_FC/877 : FAILURE in 1h 14m 52s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/4/check/DellEMC_Unity_iSCSI/7fb0a29/DellEMC_Unity_iSCSI/838 : SUCCESS in 1h 35m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"1bfde7c3cc202b6ab3e1df89987a4cbe3c495768","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-01 01:00:54.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/671235/4/check/dsvm-tempest-cisco-zonemanager-job/0e7f475/ : SUCCESS in 1h 10m 41s","accounts_in_message":[],"_revision_number":4},{"id":"6224bdea5d4676ba696903eb2c860fc1b536982b","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 01:22:20.000000000","message":"Patch Set 4:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-35-671235-4-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d5b576e7d782041a36fdf3a4111ca1c4746c9ba9","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-01 01:53:47.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/4/check/kaminario-dsvm-tempest-full-iscsi/10aca9b : SUCCESS in 1h 49m 22s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/4/check/kaminario-dsvm-tempest-full-FC/9a6473c : FAILURE in 1m 12s","accounts_in_message":[],"_revision_number":4},{"id":"bf4e775707bd9ed4e785defae24de2cba6b78c2d","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-01 01:54:24.000000000","message":"Patch Set 4:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":4},{"id":"ad3502b7ff28aedabdfbfa7da838e5e9957d1717","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-01 02:59:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/35/671235/4/check-cinder/Cinder-ISER-ISCSI/a9507af : SUCCESS in 1h 12m 30s\n- Cinder-ISER-LIO http://13.74.249.42/35/671235/4/check-cinder/Cinder-ISER-LIO/30302d4 : SUCCESS in 1h 02m 22s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"f7528a15c185673e7413677ad29595ca262163f0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 03:10:01.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/4/check/DellEMC_VNX_iSCSI/d026a49/DellEMC_VNX_iSCSI/502 : SUCCESS in 1h 42m 13s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/4/check/DellEMC_VNX_FC/d68d95f/DellEMC_VNX_FC/512 : SUCCESS in 1h 02m 37s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"78d0b13dadb1376065878ee7a79147af1c527631","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-01 06:07:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/35/671235/4/check/dsvm-tempest-zadara-driver/c5c42ed : SUCCESS in 1h 42m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bef7716744ff6a03a95f869c1300254e4a9a9ff8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-01 08:46:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/4/check/DellEMC_SC_FC/fa49607/DellEMC_SC_FC/367 : SUCCESS in 1h 08m 40s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/4/check/DellEMC_SC_iSCSI/6192e36/DellEMC_SC_iSCSI/349 : FAILURE in 2h 52m 57s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"e2e9772c363215578bdcbc6ded24d97ee29c9e43","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 09:44:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/4/check/DellEMC_XtremIO_FC/acb304c/DellEMC_XtremIO_FC/1008 : SUCCESS in 1h 27m 55s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/4/check/DellEMC_XtremIO_iSCSI/9df07e8/DellEMC_XtremIO_iSCSI/941 : FAILURE in 23m 26s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"02e0ae1110d45606dce27a566354a1db93805fae","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-01 10:52:43.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/35/671235/4/check-not-vote/ext-cinder-zuul/f551b5e : FAILURE in 1h 00m 48s","accounts_in_message":[],"_revision_number":4},{"id":"dc87731aae5511d0045573b703476ae8706ee81c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-01 16:19:01.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/35/671235/4/check/dsvm-tempest-infinibox-fc/3b76789 : SUCCESS in 1h 10m 10s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/35/671235/4/check/dsvm-tempest-infinibox-iscsi/79c511c : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"73f5d8e576d615a99ee18cec0725db20ad2997bb","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 17:02:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/4/check/DellEMC_PowerMAX_FC/a9448a1/DellEMC_PowerMAX_FC/546 : FAILURE in 12m 36s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/4/check/DellEMC_PowerMAX_iSCSI/2f0f56b/DellEMC_PowerMAX_iSCSI/531 : FAILURE in 5h 13m 13s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"8b407979f87b087e244ac35fb33ddcc7c10f74db","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-02 02:02:41.000000000","message":"Patch Set 4:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":4},{"id":"11c9c8959291e90c9cf95df29045c6c22746960e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 02:30:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/35/671235/4/check/NEC-ISCSI/1ad208b : SUCCESS in 2h 05m 18s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/35/671235/4/check/NEC-FC/ff82d16 : SUCCESS in 1h 27m 20s","accounts_in_message":[],"_revision_number":4},{"id":"246d374ec440cd64692df0e2eaf5d91ec6816a3f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 18:40:10.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/4/check/DellEMC_PowerMAX_FC/b2cc5f3/DellEMC_PowerMAX_FC/562 : FAILURE in 3h 35m 31s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/4/check/DellEMC_PowerMAX_iSCSI/81b1418/DellEMC_PowerMAX_iSCSI/537 : FAILURE in 5h 56m 59s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"a432418b720fe845369b14697dfc79bdaa3cce90","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-05 01:44:29.000000000","message":"Patch Set 4:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":4},{"id":"a5bce0c8f2fa39b4dd5b0f4022aee12f98761dd4","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-05 13:40:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/4/check/DellEMC_PowerMAX_FC/87efdd9/DellEMC_PowerMAX_FC/577 : FAILURE in 4h 11m 05s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/4/check/DellEMC_PowerMAX_iSCSI/8c66452/DellEMC_PowerMAX_iSCSI/552 : FAILURE in 5h 39m 45s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"6bc000f3ba7715653deabeabc19d83d28016728b","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-07 09:43:46.000000000","message":"Patch Set 4:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":4},{"id":"0362d20021f7e0999f6c8965d6ae05ec4e15192e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 20:50:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/4/check/DellEMC_PowerMAX_FC/16b6038/DellEMC_PowerMAX_FC/585 : FAILURE in 3h 45m 05s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/4/check/DellEMC_PowerMAX_iSCSI/4bf636b/DellEMC_PowerMAX_iSCSI/557 : SUCCESS in 5h 10m 17s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"3d69a45b44945a248771585d29b3dd00ea21dbe0","author":{"_account_id":29627,"name":"MacroSAN Volume CI","email":"466367416@qq.com","username":"hjy"},"date":"2019-08-08 02:16:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-isicsi-driver http://120.77.149.115/35/671235/4/check/cinder-isicsi-driver/e5f751c : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":4},{"id":"64caf27f42c17cb4b34e59e1c65e0228b47ce721","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-08 09:47:56.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"aa3764629a9f133acf2b8efe86c90b69a18bc191","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-08 09:49:58.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"c9d05fb5777f5493934693f64b307e5b3341fd8b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-08 11:05:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/5/check/fujitsu-eternus-dx-iscsi/f020aa0 : FAILURE in 42s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/5/check/fujitsu-eternus-dx-fc/b54db6f : FAILURE in 56s","accounts_in_message":[],"_revision_number":5},{"id":"99cfed5465f0857be57a2160043eabc8f1bb504a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-08 11:47:44.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/5/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f51248a : SUCCESS in 54m 12s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/5/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/d0f15ae : FAILURE in 5s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"1379b63b1f8f6bcc26cb135fda40e9f7d739e455","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-08 12:03:48.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1912 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"168e0e558e319f67d827a705f7b120a829955d20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 13:09:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/35/671235/5/check/openstack-tox-lower-constraints/2d0552b/ : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://logs.opendev.org/35/671235/5/check/openstack-tox-pep8/c999df1/ : SUCCESS in 8m 31s\n- openstack-tox-py27 https://logs.opendev.org/35/671235/5/check/openstack-tox-py27/83b3dd8/ : SUCCESS in 10m 30s\n- openstack-tox-py36 https://logs.opendev.org/35/671235/5/check/openstack-tox-py36/787b021/ : SUCCESS in 9m 29s\n- openstack-tox-py37 https://logs.opendev.org/35/671235/5/check/openstack-tox-py37/3ff09c9/ : SUCCESS in 9m 06s\n- openstack-tox-docs https://logs.opendev.org/35/671235/5/check/openstack-tox-docs/b1e0de7/html/ : SUCCESS in 9m 13s\n- tempest-full https://logs.opendev.org/35/671235/5/check/tempest-full/3c50121/ : SUCCESS in 1h 52m 35s\n- neutron-grenade https://logs.opendev.org/35/671235/5/check/neutron-grenade/8e7ad0c/ : SUCCESS in 1h 03m 57s\n- grenade-py3 https://logs.opendev.org/35/671235/5/check/grenade-py3/3b96f52/ : SUCCESS in 1h 05m 01s\n- tempest-integrated-storage https://logs.opendev.org/35/671235/5/check/tempest-integrated-storage/7bf0c5a/ : SUCCESS in 1h 32m 01s\n- build-openstack-releasenotes https://logs.opendev.org/35/671235/5/check/build-openstack-releasenotes/e31c076/html/ : SUCCESS in 15m 29s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/35/671235/5/check/cinder-tempest-dsvm-lvm-lio-barbican/267373c/ : SUCCESS in 1h 33m 24s\n- cinder-tox-bandit-baseline https://logs.opendev.org/35/671235/5/check/cinder-tox-bandit-baseline/63e8fe3/ : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/35/671235/5/check/openstack-tox-functional-py36/01c681c/ : SUCCESS in 4m 42s\n- cinder-rally-task https://logs.opendev.org/35/671235/5/check/cinder-rally-task/f0ae9e4/ : SUCCESS in 46m 27s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/35/671235/5/check/openstack-tox-pylint/937f92d/ : SUCCESS in 4m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/35/671235/5/check/cinder-plugin-ceph-tempest/0025ff8/ : SUCCESS in 1h 22m 00s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/35/671235/5/check/openstack-tox-functional/1969885/ : SUCCESS in 5m 20s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/35/671235/5/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/5e75122/ : SUCCESS in 1h 08m 16s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/35/671235/5/check/legacy-tempest-dsvm-lvm-multibackend/403302e/ : SUCCESS in 1h 17m 46s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/35/671235/5/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/3c7361a/ : SUCCESS in 1h 25m 57s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/35/671235/5/check/tempest-slow-py3/531d00d/ : SUCCESS in 2h 31m 38s","accounts_in_message":[],"_revision_number":5},{"id":"008d4c13301e1e1526697f99ba4b7a000dd70886","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-08 13:11:07.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c85d778ae92ce5e9fe6c240ad6973edbdabdf2b5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-08 13:11:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/35/671235/5/upstream-solidfire-check/cinder-solidfire-iSCSI/543c61e/ : NOT_REGISTERED\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"08479c3331d54f0a60c7f2f6c92b47c67138f226","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-08 13:13:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/35/671235/5/check/dsvm-tempest-infinibox-fc/ab37131 : SUCCESS in 1h 10m 16s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/35/671235/5/check/dsvm-tempest-infinibox-iscsi/acfe4a3 : SUCCESS in 1h 10m 36s","accounts_in_message":[],"_revision_number":5},{"id":"43fb63486091d9b1f38dd09d032b28175b7a58d5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 14:39:02.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/5/check/dsvm-tempest-kvm/91f78ba : FAILURE in 37m 05s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/5/check/dsvm-tempest-exe/3d5116d : FAILURE in 1m 16s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"46ecf94899bf053a8fa1788ef7e1b573ce444aa2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-08 15:10:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/5/iscsi/ : SUCCESS in 1h 11m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/5/smb/ : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":5},{"id":"214a5b1c12dc4210374a9204240eef32c34841d5","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-08 15:24:57.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/35/671235/5/check/synology-dsm-cinder-iscsi/9e7a367 : FAILURE in 38m 31s","accounts_in_message":[],"_revision_number":5},{"id":"d4d0cb521ba4d3652e48064a38715514decd5d7e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-08 15:41:30.000000000","message":"Patch Set 5:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/339356d : SUCCESS in 1h 54m 28s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/5/check/3par-fc-driver-master-client-pip-ssa02-dsvm/121584c : FAILURE in 2h 00m 15s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/35/671235/5/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/28d34df : FAILURE in 1h 32m 24s","accounts_in_message":[],"_revision_number":5},{"id":"6159ef57ed323c64bdfcac93a7d8c5fa453f21eb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-08 16:13:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.161.150/logs/35/671235/5/upstream-check/cinder-cDOT-iSCSI/ca7e450/ : SUCCESS in 1h 02m 55s\n- cinder-cDOT-NFS http://13.56.161.150/logs/35/671235/5/upstream-check/cinder-cDOT-NFS/44b37ee/ : SUCCESS in 3h 02m 43s\n- cinder-cDOT-FCP http://13.56.161.150/logs/35/671235/5/upstream-check/cinder-cDOT-FCP/e0be4d8/ : SUCCESS in 1h 56m 01s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"22175c4aab7f45225cb9ed1be4b415a89797b09d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-08 16:43:24.000000000","message":"Patch Set 5:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-35-671235-5/ns5_iscsi : FAILURE in 51m 18s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-35-671235-5/ns5_nfs : FAILURE in 50m 52s","accounts_in_message":[],"_revision_number":5},{"id":"f5c131669a91830cba25d37f382e097c897969f5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 17:36:54.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/35/671235/5/check/NEC-ISCSI/ccfd8db : SUCCESS in 2h 58m 27s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/35/671235/5/check/NEC-FC/909ebdf : FAILURE in 3h 12m 28s","accounts_in_message":[],"_revision_number":5},{"id":"018abc06d7c7303880ccd629059ae9d13cadb736","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-08 17:41:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/35/671235/5/check-cinder/Cinder-ISER-ISCSI/ea39e18 : SUCCESS in 1h 06m 48s\n- Cinder-ISER-LIO http://13.74.249.42/35/671235/5/check-cinder/Cinder-ISER-LIO/2acd829 : SUCCESS in 1h 04m 39s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"f2a10ebe25a652cc84c471b463aee7398b884250","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-08 17:57:24.000000000","message":"Patch Set 5:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-671235-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"921f3a7a9474a656bf1a0b11f369802c948d1f90","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-08 18:11:14.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/5/check/DellEMC_Unity_FC/ba3efcf/DellEMC_Unity_FC/955 : SUCCESS in 1h 10m 03s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/5/check/DellEMC_Unity_iSCSI/83ab7b1/DellEMC_Unity_iSCSI/913 : SUCCESS in 1h 27m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a5f818294d278545f9d9103b175c02238aad91b4","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-08 18:20:49.000000000","message":"Patch Set 5:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/35/2475 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e2aef6740583066e01765c8f146d53e4f12b11c6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 18:37:10.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/5/check/DellEMC_XtremIO_FC/7c4c106/DellEMC_XtremIO_FC/1085 : SUCCESS in 1h 27m 43s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/5/check/DellEMC_XtremIO_iSCSI/bb46c9c/DellEMC_XtremIO_iSCSI/1007 : SUCCESS in 1h 35m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"c691e04d428ec2350c01babc280f4ac2a78f31e8","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-08 18:57:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/5/check/kaminario-dsvm-tempest-full-iscsi/ad9915c : SUCCESS in 1h 32m 20s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/5/check/kaminario-dsvm-tempest-full-FC/b331a80 : SUCCESS in 1h 27m 27s","accounts_in_message":[],"_revision_number":5},{"id":"7cee7c72fdc9f1336b7e169b9d8748d95776ce5a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-08 20:16:21.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/5/check/DellEMC_SC_FC/184ce89/DellEMC_SC_FC/408 : SUCCESS in 1h 10m 07s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/5/check/DellEMC_SC_iSCSI/03ff173/DellEMC_SC_iSCSI/390 : SUCCESS in 1h 49m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"d113c0d74e33f01313e34b835ff1be4e6e4ffab5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 20:26:20.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/35/671235/5/check/DellEMC_VxFlexOS/6f60218/DellEMC_VxFlexOS/530 : SUCCESS in 1h 55m 21s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"bbdcb463e9195da59f8c5f80eeb9cae0348368fb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 20:39:06.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/5/check/DellEMC_VNX_iSCSI/ad63e9f/DellEMC_VNX_iSCSI/542 : FAILURE in 2h 02m 20s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/5/check/DellEMC_VNX_FC/b1a339b/DellEMC_VNX_FC/552 : SUCCESS in 57m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"88ad4099200a5c3861a571cae74d0e9226e96cbb","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-08 21:03:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/35/671235/5/check/dsvm-tempest-zadara-driver/adb0d16 : SUCCESS in 1h 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d5fffb74e3786742f4b995c6ebb991b24990292b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-08 21:25:05.000000000","message":"Patch Set 5:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/671235-5/HPMSA-FC : FAILURE in 47:08.53\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/671235-5/HPMSA-iSCSI : SUCCESS in 41:57.80","accounts_in_message":[],"_revision_number":5},{"id":"f8792104d40199db53f4e465ff08fe50421fc3d5","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-09 04:36:01.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671235/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8468387aee224316aded82704e8b95244a1a3be9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-09 05:39:31.000000000","message":"Patch Set 5:\n\nBuild Successful. \nFor 3rd party ci contact info:\n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/334/671235/5/check/infortrend-cinder-volume-fc/01ec61bbb17c984a18bdd6f8f2667a193ac20c65 : SUCCESS\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/377/671235/5/check/infortrend-cinder-volume-iscsi/01ec61bbb17c984a18bdd6f8f2667a193ac20c65 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"125f9cdc1c6a38bda89d639cef3deacfb02f1302","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-08-09 06:53:06.000000000","message":"Patch Set 5:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/671235-5/Lenovo-FC : FAILURE in 1:18:23\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/671235-5/Lenovo-iSCSI : FAILURE in 2:36:52","accounts_in_message":[],"_revision_number":5},{"id":"ce131f42509bbd9d124f7259328b0933a740d2f8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-09 08:22:21.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/5/check/tempest-dsvm-full-xenial/9e22ab2/ : SUCCESS in 1h 25m 30s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/5/check/devstack-plugin-ceph-dsvm-xenial/359a5af/ : SUCCESS in 1h 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b95de585527a8501fa02eeefc139df2dbb72bf8b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-09 12:15:57.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/35/671235/5/check-not-vote/ext-cinder-zuul/0cf2ef3 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":5},{"id":"44363c625d240146a4bbf91c6a5f03c909be0e70","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-09 21:12:37.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"ca754d79a2dc722e16381d5a670890edba51ac86","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-09 21:13:12.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"df5adbefa596af50456c7deb6d9b421dbde49536","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-09 22:05:35.000000000","message":"Patch Set 6:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":6},{"id":"545b84a2e5a8dcca6056d68fa5d197fb597a4cc7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-09 22:13:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/6/iscsi/ : SUCCESS in 1h 00m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/6/smb/ : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":6},{"id":"2485762778de7d08fecb6fc43e7b6be9a391b1d7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-09 22:43:11.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/6/check/fujitsu-eternus-dx-iscsi/87f84da : FAILURE in 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/6/check/fujitsu-eternus-dx-fc/f11378a : FAILURE in 1m 02s","accounts_in_message":[],"_revision_number":6},{"id":"1a8825d9270401d1ec6e6a3d153bf5964332cb95","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-09 23:32:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/6/check/dsvm-tempest-kvm/33ac5b1 : FAILURE in 35m 08s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/6/check/dsvm-tempest-exe/f710383 : FAILURE in 35m 38s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"3246a998bacac15498be2d5aee8762ce035e1f6c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-09 23:45:32.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/6/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f064087 : FAILURE in 1h 22m 46s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/6/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/9cbc7ac : SUCCESS in 1h 18m 45s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"9e0b4838395511bd7289368ea6f454fda480e478","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-09 23:52:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/35/671235/6/check/dsvm-tempest-infinibox-fc/aed5cd2 : SUCCESS in 55m 18s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/35/671235/6/check/dsvm-tempest-infinibox-iscsi/17f8d47 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":6},{"id":"d02894d8e3a0e639fd9a9397626d42ae78924fe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 03:48:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/35/671235/6/check/openstack-tox-lower-constraints/3284901/ : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://logs.opendev.org/35/671235/6/check/openstack-tox-pep8/fb84342/ : SUCCESS in 8m 44s\n- openstack-tox-py27 https://logs.opendev.org/35/671235/6/check/openstack-tox-py27/ae75268/ : SUCCESS in 9m 28s\n- openstack-tox-py36 https://logs.opendev.org/35/671235/6/check/openstack-tox-py36/6b5bb5f/ : SUCCESS in 9m 34s\n- openstack-tox-py37 https://logs.opendev.org/35/671235/6/check/openstack-tox-py37/b90a6bc/ : SUCCESS in 9m 05s\n- openstack-tox-docs https://logs.opendev.org/35/671235/6/check/openstack-tox-docs/1516cb1/html/ : SUCCESS in 7m 14s\n- tempest-full https://logs.opendev.org/35/671235/6/check/tempest-full/99e448c/ : SUCCESS in 1h 20m 13s\n- neutron-grenade https://logs.opendev.org/35/671235/6/check/neutron-grenade/64d4d7f/ : SUCCESS in 54m 31s\n- grenade-py3 https://logs.opendev.org/35/671235/6/check/grenade-py3/6d36584/ : SUCCESS in 1h 07m 33s\n- tempest-integrated-storage https://logs.opendev.org/35/671235/6/check/tempest-integrated-storage/9097226/ : SUCCESS in 1h 17m 42s\n- build-openstack-releasenotes https://logs.opendev.org/35/671235/6/check/build-openstack-releasenotes/f410587/html/ : SUCCESS in 15m 18s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/35/671235/6/check/cinder-tempest-dsvm-lvm-lio-barbican/4268c89/ : FAILURE in 2h 48m 46s\n- cinder-tox-bandit-baseline https://logs.opendev.org/35/671235/6/check/cinder-tox-bandit-baseline/d0bd8eb/ : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/35/671235/6/check/openstack-tox-functional-py36/8b8699b/ : SUCCESS in 6m 09s\n- cinder-rally-task https://logs.opendev.org/35/671235/6/check/cinder-rally-task/17d8482/ : SUCCESS in 42m 58s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/35/671235/6/check/openstack-tox-pylint/d581e5f/ : SUCCESS in 4m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/35/671235/6/check/cinder-plugin-ceph-tempest/51efd7f/ : SUCCESS in 1h 12m 56s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/35/671235/6/check/openstack-tox-functional/74522b3/ : SUCCESS in 6m 00s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/35/671235/6/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/08218bb/ : SUCCESS in 1h 06m 15s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/35/671235/6/check/legacy-tempest-dsvm-lvm-multibackend/8da3d42/ : SUCCESS in 1h 20m 26s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/35/671235/6/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/533a642/ : SUCCESS in 1h 27m 47s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/35/671235/6/check/tempest-slow-py3/8850973/ : SUCCESS in 2h 53m 31s","accounts_in_message":[],"_revision_number":6},{"id":"83f96ac8a1ee445a1b7be3b342ed7e1176912c34","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-10 06:12:08.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/6/check/tempest-dsvm-full-xenial/13e33d9/ : SUCCESS in 1h 31m 55s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/6/check/devstack-plugin-ceph-dsvm-xenial/4b47d03/ : SUCCESS in 1h 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f6ff65a429be5585b1fb85826eea67161181cea2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-10 08:26:06.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/671235/6/check/dsvm-tempest-cisco-zonemanager-job/a271b4d/ : FAILURE in 1h 07m 02s","accounts_in_message":[],"_revision_number":6},{"id":"7f4dc5df3a1ff31b1b99ffe51abd0a4ba3e08cd6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-10 09:04:21.000000000","message":"Patch Set 5:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/345/671235/5/check/infortrend-cinder-volume-fc/01ec61bbb17c984a18bdd6f8f2667a193ac20c65 : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/388/671235/5/check/infortrend-cinder-volume-iscsi/01ec61bbb17c984a18bdd6f8f2667a193ac20c65 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"3b8c9d313a939e92f127087732e80ef1b131bc32","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-10 10:13:13.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_Unity_FC/b5c4273/DellEMC_Unity_FC/981 : SUCCESS in 1h 07m 02s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_Unity_iSCSI/05f0b00/DellEMC_Unity_iSCSI/938 : SUCCESS in 1h 28m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"f225925d99d538adec016cca747db52b593477ab","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-10 11:13:33.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_XtremIO_FC/8966d84/DellEMC_XtremIO_FC/1110 : SUCCESS in 1h 27m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_XtremIO_iSCSI/e49c55b/DellEMC_XtremIO_iSCSI/1032 : SUCCESS in 1h 35m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"4f3f335a53ac43c8a4b11804813c1df9ba9b606f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-10 20:25:41.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/35/671235/6/check/NEC-ISCSI/861e5d5 : SUCCESS in 2h 06m 18s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/35/671235/6/check/NEC-FC/aa1de79 : FAILURE in 56m 10s","accounts_in_message":[],"_revision_number":6},{"id":"e760fde59b929d916ec6d2841d40dced76124edc","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-11 04:38:21.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671235/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"1a9379bbffdf3e0cb3af3d5f757b030075cc4f15","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-11 22:45:20.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_FC/70117b2/DellEMC_PowerMAX_FC/607 : SUCCESS in 6h 01m 23s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_iSCSI/6328087/DellEMC_PowerMAX_iSCSI/576 : FAILURE in 8h 03m 58s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"5d8862e005bf045a9c8f549dbbc865fb62b25959","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-12 08:20:00.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c83596c2d7613e2177f949251543f9aef8ea6efb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-12 08:25:50.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"f59c927efce90a95de8006cb4cdb10e42e8a7745","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-12 09:26:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/6/iscsi/ : SUCCESS in 1h 05m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/6/smb/ : SUCCESS in 1h 06m 07s","accounts_in_message":[],"_revision_number":6},{"id":"b960abe05687529064ab6f04d9592e8673fe9624","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-12 09:48:44.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/6/check/tempest-dsvm-full-xenial/d549bd3/ : SUCCESS in 1h 24m 54s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/6/check/devstack-plugin-ceph-dsvm-xenial/e892546/ : SUCCESS in 1h 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"00bc2504a824fec78f9883999570cabf1492be97","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 10:40:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/6/check/dsvm-tempest-kvm/7680b2b : FAILURE in 34m 52s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/6/check/dsvm-tempest-exe/6f23571 : FAILURE in 35m 42s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"e775ff0e0afa46252122491d6e69ab76f7a24fea","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-12 10:44:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/6/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/484bc8e : SUCCESS in 1h 11m 38s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/6/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/c519aaf : SUCCESS in 1h 10m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"587344a81b9f2128a5e526b3624046c7ec59103d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 10:55:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/35/671235/6/check/openstack-tox-lower-constraints/cb7ad19/ : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://logs.opendev.org/35/671235/6/check/openstack-tox-pep8/60a3baa/ : SUCCESS in 9m 16s\n- openstack-tox-py27 https://logs.opendev.org/35/671235/6/check/openstack-tox-py27/6d23066/ : SUCCESS in 10m 49s\n- openstack-tox-py36 https://logs.opendev.org/35/671235/6/check/openstack-tox-py36/4da0744/ : SUCCESS in 10m 29s\n- openstack-tox-py37 https://logs.opendev.org/35/671235/6/check/openstack-tox-py37/3b9f11b/ : SUCCESS in 9m 13s\n- openstack-tox-docs https://logs.opendev.org/35/671235/6/check/openstack-tox-docs/77084ed/html/ : SUCCESS in 9m 02s\n- tempest-full https://logs.opendev.org/35/671235/6/check/tempest-full/c5c1dba/ : SUCCESS in 1h 38m 42s\n- neutron-grenade https://logs.opendev.org/35/671235/6/check/neutron-grenade/7705417/ : SUCCESS in 55m 51s\n- grenade-py3 https://logs.opendev.org/35/671235/6/check/grenade-py3/ebdeb9c/ : SUCCESS in 55m 24s\n- tempest-integrated-storage https://logs.opendev.org/35/671235/6/check/tempest-integrated-storage/eb5c41b/ : SUCCESS in 1h 11m 32s\n- build-openstack-releasenotes https://logs.opendev.org/35/671235/6/check/build-openstack-releasenotes/31dd19a/html/ : SUCCESS in 15m 19s\n- cinder-tempest-dsvm-lvm-lio-barbican https://logs.opendev.org/35/671235/6/check/cinder-tempest-dsvm-lvm-lio-barbican/0387302/ : SUCCESS in 1h 29m 30s\n- cinder-tox-bandit-baseline https://logs.opendev.org/35/671235/6/check/cinder-tox-bandit-baseline/a14754b/ : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://logs.opendev.org/35/671235/6/check/openstack-tox-functional-py36/46d43d9/ : SUCCESS in 5m 23s\n- cinder-rally-task https://logs.opendev.org/35/671235/6/check/cinder-rally-task/18526b1/ : SUCCESS in 38m 08s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/35/671235/6/check/openstack-tox-pylint/fcecc38/ : SUCCESS in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://logs.opendev.org/35/671235/6/check/cinder-plugin-ceph-tempest/426d283/ : SUCCESS in 1h 08m 14s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/35/671235/6/check/openstack-tox-functional/9dfdc95/ : SUCCESS in 5m 18s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://logs.opendev.org/35/671235/6/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/d39eb0f/ : SUCCESS in 1h 09m 35s\n- legacy-tempest-dsvm-lvm-multibackend https://logs.opendev.org/35/671235/6/check/legacy-tempest-dsvm-lvm-multibackend/1cd7a18/ : SUCCESS in 1h 34m 08s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://logs.opendev.org/35/671235/6/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/487944d/ : SUCCESS in 1h 31m 29s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/35/671235/6/check/tempest-slow-py3/3e8c473/ : SUCCESS in 2h 24m 21s","accounts_in_message":[],"_revision_number":6},{"id":"48e4d2712118ecf33ef844fd5f45868f3453746a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-12 11:05:20.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/35/671235/6/check-not-vote/ext-cinder-zuul/a5d9c22 : FAILURE in 2m 11s","accounts_in_message":[],"_revision_number":6},{"id":"a1951f2de6dd16b423a7680b03f3875b7c747b79","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-12 12:07:12.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/35/671235/6/upstream-solidfire-check/cinder-solidfire-iSCSI/c011006/ : POST_FAILURE in 1h 09m 48s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"93e015db905d8b175d6685d92fa4dbd86df34c0c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-12 12:27:06.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.161.150/logs/35/671235/6/upstream-check/cinder-cDOT-iSCSI/d54fc36/ : POST_FAILURE in 1h 30m 35s\n- cinder-cDOT-NFS http://13.56.161.150/logs/35/671235/6/upstream-check/cinder-cDOT-NFS/5861c15/ : POST_FAILURE in 1h 02m 22s\n- cinder-cDOT-FCP http://13.56.161.150/logs/35/671235/6/upstream-check/cinder-cDOT-FCP/e18e122/ : POST_FAILURE in 1h 19m 55s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"f5aeb3c75e0bb1f9f4154d8cc4014ec148346194","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-12 12:49:10.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c4e67642ee65c025e9a4a0feab72f1f04443896d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-12 13:06:20.000000000","message":"Patch Set 6:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a9d8099 : SUCCESS in 2h 02m 27s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/6/check/3par-fc-driver-master-client-pip-ssa02-dsvm/7410cd0 : SUCCESS in 2h 07m 46s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/35/671235/6/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/cb9df69 : FAILURE in 1h 20m 14s","accounts_in_message":[],"_revision_number":6},{"id":"08ae5f6656573e3a5041e8c8146c7eb62ab47a37","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-12 13:16:14.000000000","message":"Patch Set 6:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-671235-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"e66db8f8a9f1dcb4d4f67e1cea959a69ebfb11b7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-12 13:44:09.000000000","message":"Patch Set 6:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-35-671235-6/ns5_iscsi : FAILURE in 52m 19s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-35-671235-6/ns5_nfs : FAILURE in 56m 13s","accounts_in_message":[],"_revision_number":6},{"id":"b2568011e1db06305155808095895db50dca7840","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-12 14:00:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/35/671235/6/check/synology-dsm-cinder-iscsi/79ecfab : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":6},{"id":"3e1b50ff3040dce1f94e7a5a4724ad2ecd381e2e","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-12 14:52:24.000000000","message":"Patch Set 6:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/35/2509 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"9534108f875f02b7302b3b2cb2a4d5a0986ab6f7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 15:24:49.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_XtremIO_FC/7864062/DellEMC_XtremIO_FC/1126 : SUCCESS in 1h 29m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_XtremIO_iSCSI/b1a32db/DellEMC_XtremIO_iSCSI/1048 : SUCCESS in 1h 30m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"428e0796f3d3ad313bac2399e267e24af7d92b76","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-12 15:39:01.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_Unity_FC/0744d7c/DellEMC_Unity_FC/999 : FAILURE in 1h 07m 06s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_Unity_iSCSI/36a6d4a/DellEMC_Unity_iSCSI/954 : SUCCESS in 1h 36m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"5f237a9f9d50a6376bdb4c79d1f5dabbb479d5a4","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-12 15:59:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/6/check/kaminario-dsvm-tempest-full-iscsi/87c3680 : SUCCESS in 1h 43m 36s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/6/check/kaminario-dsvm-tempest-full-FC/0443c3b : SUCCESS in 1h 34m 00s","accounts_in_message":[],"_revision_number":6},{"id":"fbb213d06506097676c5d4cbeee16deedcf0ac8c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 16:08:36.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_VNX_iSCSI/dde775b/DellEMC_VNX_iSCSI/572 : SUCCESS in 1h 35m 42s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_VNX_FC/636e20a/DellEMC_VNX_FC/583 : SUCCESS in 1h 07m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"fc8655205d6afc12d49110d8a113c5f14370fb50","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-12 16:22:05.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_SC_FC/d90bad7/DellEMC_SC_FC/438 : SUCCESS in 1h 07m 32s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_SC_iSCSI/d98cfa0/DellEMC_SC_iSCSI/419 : SUCCESS in 1h 57m 05s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"170aa5f2c4fbec3b23ebaac65b61e7348346a228","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 16:30:19.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/35/671235/6/check/DellEMC_VxFlexOS/780868c/DellEMC_VxFlexOS/559 : SUCCESS in 1h 54m 42s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"819435c8dfcc04a70ffecc1247ddaf386b50c8c0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-12 18:16:45.000000000","message":"Patch Set 6:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/671235-6/HPMSA-FC : SUCCESS in 45:20.61\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/671235-6/HPMSA-iSCSI : SUCCESS in 46:40.40","accounts_in_message":[],"_revision_number":6},{"id":"093e79345ff37b4af92253283d2b0aa7b08bb1aa","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-12 20:04:40.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671235/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"200eb0fc046db6b413750801676b81d3e7912e6f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-13 00:01:55.000000000","message":"Patch Set 6:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/439/671235/6/check/infortrend-cinder-volume-iscsi/c7d699efcdc8205e7bbe3409ae1a896140a341bd : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"944de550ad300bfb455baab666f53aae5749be11","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-13 07:19:24.000000000","message":"Patch Set 6:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/428/671235/6/check/infortrend-cinder-volume-fc/c7d699efcdc8205e7bbe3409ae1a896140a341bd : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"ad21b6a54d7008261e00d0abacd0187e15e00c9f","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-13 08:24:17.000000000","message":"Patch Set 6:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":6},{"id":"0b6c67fb300f30ea77ee097f641350e6eb61660f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-13 23:18:58.000000000","message":"Patch Set 6:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/454/671235/6/check/infortrend-cinder-volume-iscsi/c7d699efcdc8205e7bbe3409ae1a896140a341bd : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"29fe92b3ac0e10d199c179836c7755529bf798d5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-14 00:16:02.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_FC/8404fb6/DellEMC_PowerMAX_FC/617 : FAILURE in 5h 52m 30s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_iSCSI/fa285f1/DellEMC_PowerMAX_iSCSI/596 : FAILURE in 8h 07m 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"179e4edf1ef9151f5f9cb454fca34f13ed2d18a5","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-15 04:43:26.000000000","message":"Patch Set 6:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":6},{"id":"fd9c2c223871c149f0ac9db65704e0cd81b9c056","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-15 20:40:18.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_FC/2e03562/DellEMC_PowerMAX_FC/625 : FAILURE in 5h 47m 34s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_iSCSI/ee10aa8/DellEMC_PowerMAX_iSCSI/602 : FAILURE in 8h 06m 35s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"15130babf264f7c32e905b2a68ead4aaa4145221","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-16 07:46:10.000000000","message":"Patch Set 6:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":6},{"id":"e5bfc7a97035dce0a2ee1b0d5812938f5eb7198f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 23:22:26.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_FC/8f3fadf/DellEMC_PowerMAX_FC/630 : FAILURE in 5h 49m 31s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_iSCSI/d8762f3/DellEMC_PowerMAX_iSCSI/606 : FAILURE in 7h 57m 55s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"d1283e9fd04f82785d0b9f13da485276e9699371","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-20 01:42:35.000000000","message":"Patch Set 6:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":6},{"id":"facdfc44d11f8b6852cf1f92ffa80008e7aa2a18","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 17:45:35.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_FC/cf4adf6/DellEMC_PowerMAX_FC/638 : FAILURE in 5h 49m 19s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_iSCSI/bbd044d/DellEMC_PowerMAX_iSCSI/614 : FAILURE in 8h 24m 48s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"5a06bfb6e8acb05bdf365f80b3bbcbe38d88274d","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-21 01:59:17.000000000","message":"Patch Set 6:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":6},{"id":"3f1c20216640b7d0e3ec3e662b79a1468c1a21ef","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-21 06:38:25.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_FC/2a83ffb/DellEMC_PowerMAX_FC/642 : FAILURE in 3h 43m 00s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_iSCSI/6a047d8/DellEMC_PowerMAX_iSCSI/617 : FAILURE in 4h 38m 44s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"1d9790af50b42d669320da2afdbd27eda15b6eda","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-22 05:38:08.000000000","message":"Patch Set 6:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":6},{"id":"3b7a1a053aa551711fa6345323e38069d824642f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-22 10:44:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_FC/cc7d8f7/DellEMC_PowerMAX_FC/650 : FAILURE in 4h 53m 59s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/6/check/DellEMC_PowerMAX_iSCSI/f71883b/DellEMC_PowerMAX_iSCSI/623 : FAILURE in 5h 06m 25s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"33f108330bf98f9779e513365cfba5e8805b88d2","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-26 13:30:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3909d2143db80850a58cd9504fd02e107e4dff00","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-26 13:43:51.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1676 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1711 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1878 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"718b582ea9bc493ca4c8e53c870f18abae8ae136","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-08-26 14:24:03.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"83bf0c6c0f14f7c82f5024333368969692711db9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-26 14:30:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/7/check/fujitsu-eternus-dx-iscsi/934e0dc : FAILURE in 58m 47s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/7/check/fujitsu-eternus-dx-fc/b2e9c6b : FAILURE in 2m 01s","accounts_in_message":[],"_revision_number":7},{"id":"cdd9a6669c5c9b808781496d970b00b49d637e6e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-26 14:44:17.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/7/iscsi/ : SUCCESS in 1h 13m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/7/smb/ : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":7},{"id":"a0b1027a911d50a1978723df35715dd786c2a8bd","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-26 14:47:01.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/7/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/4eb1e13 : SUCCESS in 1h 08m 32s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/7/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/6d020a4 : SUCCESS in 1h 15m 46s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"1b420fed5d68996ea4619649be469fe2d1cb601f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 15:00:56.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/35/671235/7/check/NEC-ISCSI/8ab5d1c : FAILURE in 49m 52s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/35/671235/7/check/NEC-FC/fa47f12 : FAILURE in 53m 34s","accounts_in_message":[],"_revision_number":7},{"id":"6c1cdc5cced1eab86b0c709af4c928e143240b01","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 15:09:20.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/7/check/dsvm-tempest-kvm/12ef212 : FAILURE in 36m 21s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/7/check/dsvm-tempest-exe/bb66d7f : FAILURE in 36m 57s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"338e79c5cab4ca538687084a3008c0e81057ec17","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-26 15:59:32.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/7/check/DellEMC_Unity_FC/aac152d/DellEMC_Unity_FC/1112 : SUCCESS in 1h 04m 44s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/7/check/DellEMC_Unity_iSCSI/ae758cd/DellEMC_Unity_iSCSI/1061 : SUCCESS in 1h 25m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"20fada0715354f9fab4e5134485f7793fbc03420","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:00:04.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0316fae35313475b9b395182ac4c8af1 : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb5336f34db4dc4b3684d923791bca9 : SUCCESS in 11m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6e5571926da470487856660a8ca9bb9 : SUCCESS in 12m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e39ad1f6c9da42d7acc5b2ab5db30d45 : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe09b3bd948a44f5a22ed92d7dab73e5 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9032d8690dc6475799468fe2f37c35fd : SUCCESS in 11m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5becd2d05afa4394b50948d88f4a3912 : SUCCESS in 1h 43m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ba306c8900584b29b0781db9e80051a8 : SUCCESS in 58m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/659401900866487bbe076013b553a366 : SUCCESS in 1h 03m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/083a1ad726524e5d8a0b3a1c5a8aa82e : SUCCESS in 1h 27m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62e411a6e542483995d3ae7a676dda61 : SUCCESS in 15m 39s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d0fb793d9d6747fb9b1a82f008bf89fa : SUCCESS in 1h 40m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78a9632079454982a21b0ad134eec983 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/107c8c3661a54992b970ecb39d45438d : SUCCESS in 8m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e2d62a0f12c74d0e905e39773f710b13 : SUCCESS in 1h 05m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed28b08a8ac748afadf410faa6813b5b : SUCCESS in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a7926919c38c438ba00aab9503a65604 : FAILURE in 1h 37m 57s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e6600cd827784c34becbfa37e8a7383c : SUCCESS in 5m 35s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad4573e7723a4c72b02a34a4f7e758cb : SUCCESS in 1h 08m 33s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ee7fbd6962c84a6b850c29666cd61a17 : SUCCESS in 1h 54m 52s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/1f56669ce7c14d3889c377781ab7d53c : SUCCESS in 1h 28m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/41a0f03528d64f82a42de2e8b519c79e : SUCCESS in 2h 23m 35s","accounts_in_message":[],"_revision_number":7},{"id":"18b8dc566bee34c0acf5631748c0fa28ba3539d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-26 16:03:39.000000000","message":"Patch Set 7:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/637/671235/7/check/infortrend-cinder-volume-iscsi/aba2af8cec90408593ff913f04441847c1b1f230 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/616/671235/7/check/infortrend-cinder-volume-fc/aba2af8cec90408593ff913f04441847c1b1f230 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"825334ab7b67f857c5caf355e1904a63fc41cdaf","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-26 16:21:18.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"f22ba45886112b30be50eb0635f1e4ecc449bd7d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-26 16:25:10.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/7/check/tempest-dsvm-full-xenial/4e2bbbd/ : SUCCESS in 2h 09m 22s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/7/check/devstack-plugin-ceph-dsvm-xenial/6a405b5/ : SUCCESS in 1h 54m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7965991521099b28e30097d482f396410d1f4ff7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 16:26:15.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/7/check/DellEMC_XtremIO_FC/83b977d/DellEMC_XtremIO_FC/1252 : SUCCESS in 1h 28m 02s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/7/check/DellEMC_XtremIO_iSCSI/0cfbc93/DellEMC_XtremIO_iSCSI/1173 : SUCCESS in 1h 35m 42s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"4e14abeb98c3160432cc3ea7bec11d8dcd7a1e4b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-26 16:30:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/35/671235/7/check/dsvm-tempest-zadara-driver/57e0885 : SUCCESS in 1h 43m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7cadd8188b37ca765d1025d097ff361976a1e05f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-26 16:50:25.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671235/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"94e152dc37425e4b498c993406d4f85f5fead136","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-26 16:56:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/35/671235/7/check/dsvm-tempest-infinibox-fc/9f6ab9f : SUCCESS in 55m 46s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/35/671235/7/check/dsvm-tempest-infinibox-iscsi/83f3f55 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":7},{"id":"ce35c387acba647f42b8e679eb1b6d8212cb407a","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-26 17:03:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/35/671235/7/check/synology-dsm-cinder-iscsi/52791af : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":7},{"id":"651d47e639a74e0a1fa1eec7cc69c6fd71db4913","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-26 17:04:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/35/671235/7/upstream-solidfire-check/cinder-solidfire-iSCSI/0526028/ : SUCCESS in 1h 02m 10s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"3f3802ada6ab23e4db8a8be20e257287b4092d69","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-26 17:11:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/35/671235/7/upstream-check/cinder-cDOT-iSCSI/f126eb2/ : SUCCESS in 54m 37s\n- cinder-cDOT-NFS http://13.56.159.105/logs/35/671235/7/upstream-check/cinder-cDOT-NFS/97ef1cd/ : SUCCESS in 52m 26s\n- cinder-cDOT-FCP http://13.56.159.105/logs/35/671235/7/upstream-check/cinder-cDOT-FCP/ca1af46/ : SUCCESS in 1h 10m 16s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"d0e050c63824bbab4b2b249348e21a648c92f1b7","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-26 17:22:14.000000000","message":"Patch Set 7:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/35/2594 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"0f68d3b41537c782d600faf7c9afc22308eba79f","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-26 17:35:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/7/check/kaminario-dsvm-tempest-full-iscsi/0f25a78 : SUCCESS in 1h 33m 59s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/7/check/kaminario-dsvm-tempest-full-FC/e4aab95 : SUCCESS in 1h 35m 10s","accounts_in_message":[],"_revision_number":7},{"id":"3931f4b1ca601ccac79290fe6489cc2220a22314","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-26 17:46:16.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/7/check/DellEMC_SC_FC/4ae2b05/DellEMC_SC_FC/503 : SUCCESS in 1h 16m 47s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/7/check/DellEMC_SC_iSCSI/b6db74d/DellEMC_SC_iSCSI/482 : SUCCESS in 1h 45m 45s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"480c5e35276e95fdea79e864b385acb99e5fac5e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 17:59:17.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/35/671235/7/check/DellEMC_VxFlexOS/7726fc9/DellEMC_VxFlexOS/622 : SUCCESS in 1h 58m 44s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"7be002b196bda3c89b7200f472cdfb56808f01d2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 18:21:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/7/check/DellEMC_VNX_iSCSI/dc0453d/DellEMC_VNX_iSCSI/639 : FAILURE in 2h 20m 47s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/7/check/DellEMC_VNX_FC/2e3c3ef/DellEMC_VNX_FC/654 : FAILURE in 1h 19m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"1d1d80a3dd5589d9a136a6e87925efb7dc21baef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-27 01:34:40.000000000","message":"Patch Set 7:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/648/671235/7/check/infortrend-cinder-volume-iscsi/aba2af8cec90408593ff913f04441847c1b1f230 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"c88e07fa199f2a0efb2e10a558a42622f8a536bd","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-27 02:07:21.000000000","message":"Patch Set 7:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":7},{"id":"2208e5461593d0209293263cfb7d4c88bba401f0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-27 09:51:35.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/7/check/DellEMC_Unity_FC/7809c39/DellEMC_Unity_FC/1123 : SUCCESS in 1h 13m 47s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/7/check/DellEMC_Unity_iSCSI/38b6029/DellEMC_Unity_iSCSI/1072 : SUCCESS in 1h 27m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"b59a77bb7b1f8fec876004ee55413caa76665dbf","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-27 17:34:48.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/7/check/DellEMC_PowerMAX_FC/7e404bb/DellEMC_PowerMAX_FC/680 : SUCCESS in 5h 44m 13s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/7/check/DellEMC_PowerMAX_iSCSI/33aabbd/DellEMC_PowerMAX_iSCSI/646 : FAILURE in 7h 46m 13s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"d80672f82a2a2c76981b1277ab9e6f7f32bfadbe","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-29 02:39:05.000000000","message":"Patch Set 7:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":7},{"id":"867c5e7661afc384c8733ec86027b3eb996f2ee0","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-29 13:22:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c90805bb0cb1802bacf87b593ac3b908ebfb9198","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2019-08-29 13:28:33.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1904 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1716 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1737 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"50599a3807b0ea2ce004d606b241cf30860e8463","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-29 13:39:56.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"0358f4dd9e220359257e4f9963e43e5e696a1e2e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-29 14:14:22.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/35/671235/8/check/NEC-ISCSI/1811406 : FAILURE in 26m 21s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/35/671235/8/check/NEC-FC/9f9d0d5 : FAILURE in 27m 07s","accounts_in_message":[],"_revision_number":8},{"id":"7e8f43afa060125f89959bb2c685c783879c2f14","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-29 14:32:58.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/8/iscsi/ : SUCCESS in 1h 06m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/8/smb/ : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":8},{"id":"810119dd0574d8b7f67222acf3efae57b9d28595","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-29 14:56:12.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"008211dec887b8933bc4d9763e1af246f61044cd","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-29 15:05:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/8/check/dsvm-tempest-kvm/5824814 : FAILURE in 39m 28s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/8/check/dsvm-tempest-exe/b4e1899 : FAILURE in 38m 41s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"10bc3418cd114b1536b55f2255a84fb6c27f3438","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-29 15:12:56.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/8/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1aa8812 : FAILURE in 1h 49m 55s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/8/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/27e50a5 : SUCCESS in 1h 09m 38s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"24de04184148b1ce9d24758213dc1a434a5b0b4b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-29 15:59:33.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/8/check/tempest-dsvm-full-xenial/3392b88/ : SUCCESS in 2h 27m 51s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/8/check/devstack-plugin-ceph-dsvm-xenial/b85fa7d/ : SUCCESS in 2h 24m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"990e7a0154bab03aae49f1617526f4a1943998a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-29 16:13:03.000000000","message":"Patch Set 8:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/692/671235/8/check/infortrend-cinder-volume-iscsi/f5cab7a802f83f1c368955133ed14645b7b15578 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/670/671235/8/check/infortrend-cinder-volume-fc/f5cab7a802f83f1c368955133ed14645b7b15578 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"2111ff40031af10b7c04ef8e51739a838e75326d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-29 17:10:28.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/8/check/DellEMC_Unity_FC/d8bdf2e/DellEMC_Unity_FC/1158 : FAILURE in 1h 14m 59s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/8/check/DellEMC_Unity_iSCSI/75443de/DellEMC_Unity_iSCSI/1107 : SUCCESS in 1h 22m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"4e1f1963f768c5a7b4ce54ce44c69786ab0a3f70","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-29 17:42:50.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671235/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"d1e72c893fb6e29de1afb5ba87d5a376c682fdbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 18:13:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5cd6a3faa1b4314a92860775e10f798 : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237b6865d8db4fd899e690669954b496 : SUCCESS in 8m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ab7a1511853458ab997586454be6181 : SUCCESS in 10m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/076f058757944344911ff29900116127 : SUCCESS in 11m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0a10c8454254a5a8e89ad378c13c2a4 : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436a0e1921514194a8d04b9ce13139f2 : SUCCESS in 12m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dc9b32fa895041a1a27489e8d3cb6767 : SUCCESS in 1h 44m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0013737a5f234c5d8a4ca3bbe95447a4 : SUCCESS in 52m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c5729a58ae6f4c09939ee87fad03ee89 : SUCCESS in 1h 03m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d78a052225e405d9f9c97217caa9e39 : SUCCESS in 1h 20m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c17fc2e6e4cf4c36815ff29505f14d2f : SUCCESS in 13m 53s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/19c88bd697914811a4dc50dde0e1b4d7 : SUCCESS in 1h 41m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da8ba941fdf649d8ad0595bbfe353e95 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/957b327b022c47cf826444b9c9ffbe5a : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fe220157767d49c69ac01595f1bf81bb : SUCCESS in 44m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b1d1591ce4640d3afbf77fd13d772be : SUCCESS in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a1c7d9f0a1784aaba361fe830bdae57c : SUCCESS in 1h 20m 00s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a64c1bc38e464996a163e4c691ff2425 : SUCCESS in 6m 25s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eaad884bc36c4781961e0d125a642720 : SUCCESS in 1h 10m 23s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bd661019ee854e48ae8491a958cc7649 : SUCCESS in 1h 31m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/e9b47632ae634f3d87184de23181e48c : SUCCESS in 2h 02m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1afd0f45e486483cba3a4bab902e733a : SUCCESS in 2h 24m 25s","accounts_in_message":[],"_revision_number":8},{"id":"61e5afd52e39e18ed8dc27dd6a2a8fbb6f0d8141","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-29 18:16:57.000000000","message":"Patch Set 8:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8428542 : FAILURE in 18s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/35/671235/8/check/3par-fc-driver-master-client-pip-ssa02-dsvm/4ac06ed : FAILURE in 18s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/35/671235/8/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/eb8282a : FAILURE in 18s","accounts_in_message":[],"_revision_number":8},{"id":"afb7525918ea83deaf3861c3213af0791d40b74c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-29 18:55:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/35/671235/8/check/dsvm-tempest-zadara-driver/b1df00f : FAILURE in 1h 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"beeb7dcde0c7f79454e9c242890ca0d1c0320b88","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-29 18:56:31.000000000","message":"Patch Set 8:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-35-671235-8/ns5_iscsi : SUCCESS in 40m 35s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-35-671235-8/ns5_nfs : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":8},{"id":"503eee184b887bc59d73e7a6088a65c3addad32c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-29 19:45:35.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"53d2e78d47fe92257666e6d1262272ece5ab925e","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-08-29 19:46:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/35/671235/8/check/synology-dsm-cinder-iscsi/bd12872 : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":8},{"id":"45286284d6894a51cbae943386bb65c28dd1247b","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-29 20:03:50.000000000","message":"Patch Set 8:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/35/2629 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"bc4feb6cde511ea51d2c38a9c402026ccb4b9e0a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-29 20:06:05.000000000","message":"Patch Set 8:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-671235-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"81f5de7a7d87bbd9fbd199a39f8fc5e906911033","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-08-29 20:24:00.000000000","message":"Patch Set 8:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-35-671235-8-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"4d5e4f4f16348d757e28e7e528c49aa169ecf83c","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-29 21:53:07.000000000","message":"Patch Set 8:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/8/check/kaminario-dsvm-tempest-full-iscsi/b670d8f : FAILURE in 5m 49s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/8/check/kaminario-dsvm-tempest-full-FC/e30c3b4 : SUCCESS in 1h 40m 59s","accounts_in_message":[],"_revision_number":8},{"id":"a1b3d4c9c75517bcbb9d10d931c72028d7dbc931","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-29 22:16:31.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/8/check/DellEMC_SC_FC/d4e0aa0/DellEMC_SC_FC/539 : SUCCESS in 1h 19m 28s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/8/check/DellEMC_SC_iSCSI/545c0e8/DellEMC_SC_iSCSI/516 : SUCCESS in 1h 51m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"40001542e9c30e1ee5b1ac87c37332ed03089387","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-29 23:02:55.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/35/671235/8/check/DellEMC_VxFlexOS/102ef99/DellEMC_VxFlexOS/655 : SUCCESS in 1h 56m 44s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"0c4a76c47f8c712cd02b576219bc532036177297","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-29 23:03:03.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/8/check/DellEMC_VNX_iSCSI/8ef721b/DellEMC_VNX_iSCSI/672 : FAILURE in 2h 13m 54s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/8/check/DellEMC_VNX_FC/9d642f9/DellEMC_VNX_FC/690 : FAILURE in 1h 19m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"f7c90ebf0df6fcdc8fceb2b5be7764b7e7448900","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 00:46:17.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/8/check/DellEMC_XtremIO_FC/554d6ed/DellEMC_XtremIO_FC/1296 : SUCCESS in 1h 31m 10s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/8/check/DellEMC_XtremIO_iSCSI/1f9ad9f/DellEMC_XtremIO_iSCSI/1215 : SUCCESS in 2h 07m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"639ec8a175c1018cadb90ba3934f2fb5d3d061bd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-30 01:15:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/35/671235/8/check/dsvm-tempest-infinibox-fc/8c9fe80 : SUCCESS in 54m 48s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/35/671235/8/check/dsvm-tempest-infinibox-iscsi/bba0794 : SUCCESS in 1h 12m 41s","accounts_in_message":[],"_revision_number":8},{"id":"84c56507b457ba49251e581f9c8b9cde1ae7ebb8","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-30 09:12:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e01d882fb53ab95a3f4ca672574cd9ba482aff79","author":{"_account_id":26561,"name":"liu tracy","email":"Tracy.Liu@emc.com"},"date":"2019-08-30 09:20:54.000000000","message":"Patch Set 9:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":9},{"id":"df28a0f455c2038d874a1e82e86129d11427e391","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-30 10:25:31.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/9/check/fujitsu-eternus-dx-iscsi/4326154 : FAILURE in 3m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/9/check/fujitsu-eternus-dx-fc/4cd1ba2 : FAILURE in 1h 11m 09s","accounts_in_message":[],"_revision_number":9},{"id":"fc446e37f9ced5a0423dda45ff88377e17c1e5cb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-30 11:16:52.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/9/iscsi/ : SUCCESS in 1h 14m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/9/smb/ : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":9},{"id":"c5011ab191dc3b44ad6e8b26bde4ac6bdae58889","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-30 11:23:33.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/9/check/tempest-dsvm-full-xenial/d0956c8/ : FAILURE in 2h 02m 22s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/9/check/devstack-plugin-ceph-dsvm-xenial/26e1ec9/ : SUCCESS in 2h 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d493237cd57c7ec851fa2ba0b4626b4bdc117827","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 12:37:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a602700766b7494cbf6778720e8bbfda : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca59d6c5f0b74621a0c3a79df8c8dd9c : FAILURE in 8m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2f9e62df1bf4ca7bc0747588e053280 : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1ebd288989744c8b36eee98ad1d2287 : SUCCESS in 10m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fd3cd8b9b174ea592fe03ab15dcba0a : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b214898b1e0f45e591a5a0eb1e47b383 : SUCCESS in 9m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/49fb0702f00449c2bc6ec9c05393a41c : SUCCESS in 1h 53m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ddc3a65b641c476c9511103d2b3706bd : SUCCESS in 1h 04m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/905b63fb369d4532b064cfa32df6782b : SUCCESS in 1h 04m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/588f92892bf14ad380a557139c50f1dd : SUCCESS in 1h 27m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c6eade737fa4847a7e4e409eb52065c : SUCCESS in 18m 36s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5e6bca397c524bcaa19b3d6c9eaec9ce : SUCCESS in 1h 38m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19aac52e1fb94ce483e1b5c1f22ba5d4 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/41bbce4d3a2045b38bfee105919a145e : SUCCESS in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a42326b8cc0848758dc39f3235e77abe : SUCCESS in 46m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5507cba3c10f4f898725987c663be400 : SUCCESS in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4819274ae3504976bd199da4d108b845 : SUCCESS in 1h 09m 46s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8a62f1f13b11451ab132bb46bfa6cb65 : SUCCESS in 5m 27s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/50ce0bf7c18c4c58b520ad4ded607891 : SUCCESS in 1h 10m 36s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/28cfbd4e5a004a44a3da32d383b940c5 : SUCCESS in 1h 17m 11s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/03ec6f555ad94d7da9e75c925a2d1048 : SUCCESS in 1h 46m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/546d3ed8d3974043b4b5f3ee558c6eb0 : SUCCESS in 2h 21m 29s","accounts_in_message":[],"_revision_number":9},{"id":"7ba4080d2901fe5d9c44b2ff7f95c0ce3349f548","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-30 12:53:16.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/35/671235/9/check/dsvm-tempest-infinibox-fc/78c6761 : SUCCESS in 53m 25s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/35/671235/9/check/dsvm-tempest-infinibox-iscsi/0a8e45f : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":9},{"id":"4736d5da59d4103cd3c602b9f7c0c72a5180915f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 13:51:45.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/9/check/dsvm-tempest-kvm/b2289b6 : FAILURE in 38m 44s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/9/check/dsvm-tempest-exe/86c8144 : FAILURE in 39m 08s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"4e11932581f060368d8676f4ddeef6de2a9d9864","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-30 13:52:29.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"15eab6057250af882ea1e3ab327bbe9c63dc4762","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2019-08-30 13:53:50.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1743 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"e3bb73c53099ee9bc890738ddb4ade404ba9343b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-30 14:01:53.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"81092c6e167d7013ab70c0caea09c6130dc29b76","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-30 14:32:30.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671235/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"0dc402565785ef6c9c3d28dce97a09693a7880a2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-30 14:51:31.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/10/iscsi/ : SUCCESS in 58m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/10/smb/ : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":10},{"id":"29f934cb452fd3d1c74b2e3a96efe8bae7d6b7e7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-30 15:11:58.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/35/671235/10/check/fujitsu-eternus-dx-iscsi/186d1fa : FAILURE in 1h 07m 56s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/35/671235/10/check/fujitsu-eternus-dx-fc/32121e3 : FAILURE in 2m 22s","accounts_in_message":[],"_revision_number":10},{"id":"cab06d62c36b48080e91df0a390dc5ae8837e69a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-30 17:13:58.000000000","message":"Patch Set 9:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/714/671235/9/check/infortrend-cinder-volume-iscsi/e90b921e2249ab171478deabf0d5a9bfe929d5fd : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/692/671235/9/check/infortrend-cinder-volume-fc/e90b921e2249ab171478deabf0d5a9bfe929d5fd : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"f52ba959ff71e4952d630ea0ff07f9b7df557e0a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 17:19:24.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/10/check/dsvm-tempest-kvm/ff355b9 : FAILURE in 37m 20s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/10/check/dsvm-tempest-exe/4b7ea04 : FAILURE in 35m 26s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"f1c195421f0c0b432f6ef3149cb76e28896b3fba","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-30 17:20:00.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/10/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/5228e05 : FAILURE in 1h 49m 33s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/10/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/48c9315 : SUCCESS in 1h 16m 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"d689ba9ebffabb2aa656c73e2e9903010c37a787","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-30 17:43:46.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/10/check/tempest-dsvm-full-xenial/175cc05/ : FAILURE in 2h 10m 30s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/10/check/devstack-plugin-ceph-dsvm-xenial/7c098c2/ : FAILURE in 2h 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e22b168fb71f08a69b271295394351f0840bcf76","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-30 17:49:43.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/35/671235/10/check/dsvm-tempest-infinibox-fc/c9084b4 : SUCCESS in 1h 05m 17s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/35/671235/10/check/dsvm-tempest-infinibox-iscsi/b9017f5 : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":10},{"id":"d0259f952c83a87f073c93a606bf9d213b3a717b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-30 18:56:08.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671235/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"b94651f97db0401963a28a06ea09387794f1a308","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:03:06.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5f6895603d84d969c33e62df3db76d2 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f92bb202583f452bb9b7a0d8a256a00e : SUCCESS in 10m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69456989266d433f9916f0aa72aa5aa2 : SUCCESS in 9m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dadfec6c947452f9eef1e2a55194c50 : SUCCESS in 7m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f29f7bdc18c4c79b46470f102031beb : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a59ec29274a4b51b51d526ba6adea80 : SUCCESS in 10m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/20a891a6a1fa4455ad2fee0885332012 : FAILURE in 1h 49m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6be47f7e837c4fc7b4de5c3ae4a1ae8e : SUCCESS in 1h 01m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/99b5549b0cb94e7b96a37d5b5f887377 : SUCCESS in 1h 01m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86d67ff99dc84039819198c67f7619d2 : SUCCESS in 1h 43m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c693f78b6a2e4479a04bde0bec3da671 : SUCCESS in 20m 43s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/36cbe2e3dee746bcb2683206a6f335ba : SUCCESS in 1h 38m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/daf09e037b8d461cbc14b57fc7122738 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/644b4649022344a090ffccc32d38bd81 : SUCCESS in 5m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/da7cf949e2c44786a77fa8e734b57ca2 : SUCCESS in 46m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e13b83ebf76400d9e2d04d81d2ae274 : SUCCESS in 4m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3c8df336a27e44588cec51629d154e12 : SUCCESS in 1h 25m 25s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/16e5c555da144eecb96914fa6903672c : SUCCESS in 6m 03s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/44a5c44369db4712bfa65148b374ac7e : SUCCESS in 1h 34m 44s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7ee2b83148294d07aee4a68c9e6b1048 : SUCCESS in 1h 34m 17s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/ef61c704f419480f94e0973a094c790b : SUCCESS in 1h 36m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02cb6325aced42ccb471c850021587ff : SUCCESS in 2h 24m 05s","accounts_in_message":[],"_revision_number":10},{"id":"86ff879808e166d6b6ba3adff123f5b1b54b6573","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-30 19:19:03.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/671235/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"14ba361163ae793f92cd0cd93b8bb7509499c3ed","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-08-30 20:35:28.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"a59c733fb3ae37e6fcf7470c64bba26ead760b94","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 21:23:41.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/35/671235/10/check/NEC-ISCSI/9264144 : FAILURE in 1h 16m 15s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/35/671235/10/check/NEC-FC/dc61392 : FAILURE in 1h 20m 40s","accounts_in_message":[],"_revision_number":10},{"id":"a4b2fd7ca9a248d1c27db61c48845bc4fbdcfb5c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-30 22:04:16.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671235/10/iscsi/ : FAILURE in 1h 28m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/671235/10/smb/ : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":10},{"id":"b5ca8ce49b6ee35fef142a41760c18aeae8290d7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-30 22:14:43.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/10/check/tempest-dsvm-full-xenial/7c0c92f/ : SUCCESS in 1h 36m 53s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/671235/10/check/devstack-plugin-ceph-dsvm-xenial/6ba1188/ : SUCCESS in 1h 38m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"418a5bb8ec4c6d1a811258fc567f0cabd4bd7dd1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-30 23:05:39.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/10/check/DellEMC_Unity_FC/2ba9e7d/DellEMC_Unity_FC/1181 : FAILURE in 1h 08m 05s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/10/check/DellEMC_Unity_iSCSI/707f2d2/DellEMC_Unity_iSCSI/1127 : FAILURE in 8m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"93f03935150489094722ac501393e6ab613a7461","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 23:35:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/416e7f4754254888a1b94c05514b7c4c : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c3250bd0934e3183d666e05741e8bb : SUCCESS in 10m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a56ee204c6bc4a02aa8cbfd1dc3d25b3 : SUCCESS in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3a605ebbcca4d07b6a4028aec1b8ae1 : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73666fef017e4ca89797809f668bf023 : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058b6593de3848009a4c8fde31bf9959 : SUCCESS in 8m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4ff04ef387e34c85a0ae41debfdb909f : SUCCESS in 1h 57m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9a5e90a5839c49b3b0a0f06d0d12e67e : SUCCESS in 1h 03m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/edad5035aa70467aa6a25502caa1eab5 : SUCCESS in 1h 23m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f655b7b8a56438cad9d049f60962596 : SUCCESS in 1h 30m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9b720baebfd40b8a1fa55f8ec8f13c6 : SUCCESS in 17m 04s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3e12d90d573642a0ac686b2ef9f8bac3 : SUCCESS in 1h 50m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/531b8eb563e14efa8672eb43e42f55af : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3cee89b4e71406eabf38a3cf45b0dff : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8dde0e1e69b34c9692667851189553f3 : SUCCESS in 39m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/65ed3d757c254811b7348c247bf6954f : SUCCESS in 4m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1c20800bd5f147d789494b4c4d06a144 : SUCCESS in 1h 21m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/54032fd5437d429a95de44465e3ca9f3 : SUCCESS in 8m 27s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba8bf04ddccb48f5bb0dfa31a2b945e4 : SUCCESS in 1h 17m 44s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fe16c4e2bf44486ba674b459093d4226 : FAILURE in 1h 43m 41s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/06745558b1e648ef8c7f55dc6ae7fbbe : SUCCESS in 1h 56m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9300de184f8a4d088324ba6fe45f83dd : SUCCESS in 2h 17m 45s","accounts_in_message":[],"_revision_number":10},{"id":"8204c6cefeaf7fc74396edd1f828e7e56651b8a0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-31 01:23:06.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/35/671235/10/upstream-solidfire-check/cinder-solidfire-iSCSI/7370bf8/ : SUCCESS in 53m 46s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"c423995a6cc9368782a6a4b14b8f3092dda707cc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-31 01:42:01.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/35/671235/10/upstream-check/cinder-cDOT-iSCSI/ef60d3c/ : SUCCESS in 57m 34s\n- cinder-cDOT-NFS http://13.56.159.105/logs/35/671235/10/upstream-check/cinder-cDOT-NFS/0b982f1/ : SUCCESS in 56m 56s\n- cinder-cDOT-FCP http://13.56.159.105/logs/35/671235/10/upstream-check/cinder-cDOT-FCP/43176f2/ : FAILURE in 1h 13m 32s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"faaa8cadbd3bb567c55230455211f7e9686e763e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-31 01:46:34.000000000","message":"Patch Set 10:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-35-671235-10/ns5_iscsi : SUCCESS in 45m 22s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-35-671235-10/ns5_nfs : SUCCESS in 44m 46s","accounts_in_message":[],"_revision_number":10},{"id":"99d856567e158cf21633d3e65951169f5fd90582","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-31 02:17:29.000000000","message":"Patch Set 9:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/723/671235/9/check/infortrend-cinder-volume-iscsi/e90b921e2249ab171478deabf0d5a9bfe929d5fd : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/701/671235/9/check/infortrend-cinder-volume-fc/e90b921e2249ab171478deabf0d5a9bfe929d5fd : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"7532a110c09791d4580d93c597cadcaa5c3bb879","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-31 02:27:34.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/35/671235/10/check/DellEMC_VxFlexOS/95a56ed/DellEMC_VxFlexOS/677 : SUCCESS in 1h 57m 34s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":10},{"id":"12908564c1b9b2545524263ad7334c89bd2e8f06","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-31 02:53:05.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/10/check/dsvm-tempest-kvm/b8cde57 : FAILURE in 35m 28s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/671235/10/check/dsvm-tempest-exe/816948f : FAILURE in 37m 03s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"981797e9b4665d20e58fe133f27ade556d73c4ce","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-31 03:27:04.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/35/671235/10/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/a9201df : SUCCESS in 1h 00m 22s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/35/671235/10/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/fe460cd : SUCCESS in 1h 13m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"a89a0d1cceb9929e5fd07f1f4994e5dcec921df0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-31 03:34:11.000000000","message":"Patch Set 10:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-671235-10 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":10},{"id":"086bb177941a0186466a248dc0429999a3e2e043","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-31 04:43:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/35/671235/10/check/dsvm-tempest-zadara-driver/d97e2d4 : SUCCESS in 1h 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fdede8b83e02fbd3d195ed639068e056362a10ba","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-31 06:05:32.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/671235/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"2423678dee3bc6bc658da0fc771398c90fe3f08b","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-08-31 08:44:49.000000000","message":"Patch Set 10:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-35-671235-10-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"00895d5755dfcfe449e625e5d2f4c5f3646ed67b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-31 11:39:27.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/35/671235/10/check/DellEMC_Unity_FC/6b75344/DellEMC_Unity_FC/1191 : FAILURE in 1h 13m 52s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/35/671235/10/check/DellEMC_Unity_iSCSI/b852e26/DellEMC_Unity_iSCSI/1137 : SUCCESS in 1h 28m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"577b0d0f0b776ad749aef347a611df2c82774400","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-31 12:20:18.000000000","message":"Patch Set 10:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/733/671235/10/check/infortrend-cinder-volume-iscsi/4d968acc947fa5bc3296ebedee6b51b43786c63e : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/711/671235/10/check/infortrend-cinder-volume-fc/4d968acc947fa5bc3296ebedee6b51b43786c63e : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"3fc2c5752203fca6afc06c25de32368aa55fd8d5","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-31 14:26:25.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/35/671235/10/check/kaminario-dsvm-tempest-full-iscsi/d39d599 : SUCCESS in 1h 44m 07s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/35/671235/10/check/kaminario-dsvm-tempest-full-FC/66faa92 : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":10},{"id":"6f792c989b2000eeff882235f827c06fc889f4fd","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-31 14:31:36.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/35/671235/10/check/DellEMC_SC_FC/1141eae/DellEMC_SC_FC/561 : SUCCESS in 1h 12m 20s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/35/671235/10/check/DellEMC_SC_iSCSI/dd97fd5/DellEMC_SC_iSCSI/536 : SUCCESS in 1h 46m 54s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"5b9fdd5474e532c091b4248252c671f98bfde9b3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-31 15:32:51.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/35/671235/10/check/DellEMC_XtremIO_FC/1259dad/DellEMC_XtremIO_FC/1315 : SUCCESS in 1h 29m 00s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/35/671235/10/check/DellEMC_XtremIO_iSCSI/81bd16e/DellEMC_XtremIO_iSCSI/1230 : FAILURE in 2h 20m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"4c8590735c0b2f090fc1775b8ab991eaeb989934","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-31 21:19:51.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/35/671235/10/check/DellEMC_VNX_iSCSI/fc091c7/DellEMC_VNX_iSCSI/694 : FAILURE in 2h 53m 46s\n- DellEMC_VNX_FC http://publiclogs.emc.com/35/671235/10/check/DellEMC_VNX_FC/acb5ce1/DellEMC_VNX_FC/712 : SUCCESS in 1h 14m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"244eab19a6ea54f35a4e27b82c207b0377f6a233","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-31 21:40:29.000000000","message":"Patch Set 10:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/35/2654 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"459cc4b637f4ea1e6459cf0a4ce9bbaa0273ff07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-31 22:34:31.000000000","message":"Patch Set 10:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/744/671235/10/check/infortrend-cinder-volume-iscsi/4d968acc947fa5bc3296ebedee6b51b43786c63e : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/722/671235/10/check/infortrend-cinder-volume-fc/4d968acc947fa5bc3296ebedee6b51b43786c63e : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"be40af2a30d503c19953deb5e38c3c73d67938db","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-08-31 23:10:32.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/35/671235/10/check/DellEMC_PowerMAX_FC/2c7233a/DellEMC_PowerMAX_FC/700 : SUCCESS in 4h 37m 50s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/35/671235/10/check/DellEMC_PowerMAX_iSCSI/c12a863/DellEMC_PowerMAX_iSCSI/660 : SUCCESS in 5h 01m 24s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":10},{"id":"5383b28fffb000056c36e97acc6b9842ed36edbe","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2019-09-02 09:39:49.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c989119fb4823b6b3b839223cc80871133e858ee","author":{"_account_id":30165,"name":"Simon O\u0027Donovan","email":"simon.odonovan@dell.com","username":"odonos12"},"date":"2019-09-04 14:25:42.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"f6d3eb4cfccbfbdefd0f120062d35df74ab5d9fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-12 12:51:07.000000000","message":"Patch Set 10: Code-Review+2\n\nLooks good to me and CI is happy.","accounts_in_message":[],"_revision_number":10},{"id":"a8bbd02cff08d61085b49a4241f6bff2d89a335f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-09-12 14:40:36.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nCI passing.  Code looks ok.  I think this is good.  Thank you.","accounts_in_message":[],"_revision_number":10},{"id":"4ffe9f64f34596a8169b3aedb95b198f366c6075","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 14:40:59.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"203bbae6fce37217fd234197b2273c26abcc3ac4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 16:37:55.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79b4de738c614753baee9a68937597d8 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a5b0ecb31143ff84945b83a8fe583d : SUCCESS in 8m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf861c6a2eac4ebab7e6102f72b8b5fb : SUCCESS in 10m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87817d9d260a40aebaf4f133103ad84a : SUCCESS in 9m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc22fcdb518d4ebaab146b252afe689d : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89cacf1467794f3f95855f04c79c7cc6 : SUCCESS in 15m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d5a958c4754e48e9874ae48e3414682d : SUCCESS in 1h 54m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3579514f607f49c08ef1d08407024227 : SUCCESS in 53m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9947befaed524786b2289536d5148fc0 : SUCCESS in 59m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce4358b6af6f4ee39a530bfe96e5875d : SUCCESS in 1h 33m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16241d35343c4cec87e89b8443109625 : SUCCESS in 17m 16s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9aebc1058313443fb71327ad0307813a : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":10},{"id":"291f64eb8501c40d17fe13fa513d27b0861232cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 16:37:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"9b519328d9a26be3526daa265d8e09bd56bf3f32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 16:39:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbda91d481c4a87a2181c346caf899c : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/437dc5e9c8294d3bbfc5b9b393d88d9c : SUCCESS in 50s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"4d968acc947fa5bc3296ebedee6b51b43786c63e","revisions":{"8cc60b9e23bfcc9027953aa363517f743f8b2f5b":{"kind":"REWORK","_number":1,"created":"2019-07-17 10:17:23.000000000","uploader":{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},"ref":"refs/changes/35/671235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/1"}}},"commit":{"parents":[{"commit":"c8b8cb02d8de68761b52bb9ea2a87eabff2636af","subject":"PowerMax Driver - Train San REST Port Deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8b8cb02d8de68761b52bb9ea2a87eabff2636af"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 07:00:44.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cc60b9e23bfcc9027953aa363517f743f8b2f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cc60b9e23bfcc9027953aa363517f743f8b2f5b"}]},"branch":"refs/heads/master"},"30a2a7007324914fe2618681b3faaec172d75e3c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-18 10:16:47.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/2"}}},"commit":{"parents":[{"commit":"478b1f8ea308429709239d812e30c5d6c9833a00","subject":"PowerMax Driver - Train San REST Port Deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/478b1f8ea308429709239d812e30c5d6c9833a00"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-07-18 10:15:30.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30a2a7007324914fe2618681b3faaec172d75e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30a2a7007324914fe2618681b3faaec172d75e3c"}]},"branch":"refs/heads/master"},"fa1f2085dbb0b3adfd6d8b0437db9c8374faeb39":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-30 20:43:58.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/3"}}},"commit":{"parents":[{"commit":"8bbd33495d61e06375eea401a08497ee1c2caa24","subject":"PowerMax Driver - Train San REST Port Deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bbd33495d61e06375eea401a08497ee1c2caa24"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-07-30 20:43:03.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa1f2085dbb0b3adfd6d8b0437db9c8374faeb39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa1f2085dbb0b3adfd6d8b0437db9c8374faeb39"}]},"branch":"refs/heads/master"},"866a3dd84e6a99b18fb9ad1f997a6dfe1f52f4e4":{"kind":"REWORK","_number":4,"created":"2019-07-31 17:55:20.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/4"}}},"commit":{"parents":[{"commit":"4e889ce232dbc41f9ae9006b80f056e59a9aa10e","subject":"PowerMax Driver - Train San REST Port Removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e889ce232dbc41f9ae9006b80f056e59a9aa10e"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-07-31 17:54:57.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/866a3dd84e6a99b18fb9ad1f997a6dfe1f52f4e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/866a3dd84e6a99b18fb9ad1f997a6dfe1f52f4e4"}]},"branch":"refs/heads/master"},"01ec61bbb17c984a18bdd6f8f2667a193ac20c65":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-08 09:47:56.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/5"}}},"commit":{"parents":[{"commit":"c09e36fab564b385e9fdb255b48c33edada3a400","subject":"PowerMax Driver - Train San REST Port Removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c09e36fab564b385e9fdb255b48c33edada3a400"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-08-08 09:46:58.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/01ec61bbb17c984a18bdd6f8f2667a193ac20c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/01ec61bbb17c984a18bdd6f8f2667a193ac20c65"}]},"branch":"refs/heads/master"},"c7d699efcdc8205e7bbe3409ae1a896140a341bd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-09 21:12:37.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/6"}}},"commit":{"parents":[{"commit":"0d2f3d0d5d5d51d8a2903f10636b269f4164d6cc","subject":"PowerMax Driver - Train San REST Port Removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d2f3d0d5d5d51d8a2903f10636b269f4164d6cc"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-08-09 21:11:41.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7d699efcdc8205e7bbe3409ae1a896140a341bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7d699efcdc8205e7bbe3409ae1a896140a341bd"}]},"branch":"refs/heads/master"},"aba2af8cec90408593ff913f04441847c1b1f230":{"kind":"REWORK","_number":7,"created":"2019-08-26 13:30:25.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/7"}}},"commit":{"parents":[{"commit":"385f554743d326a23e2b5a1882691d67dfbe0594","subject":"PowerMax Driver - Miscellaneous improvements to delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/385f554743d326a23e2b5a1882691d67dfbe0594"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-08-26 13:30:03.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aba2af8cec90408593ff913f04441847c1b1f230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aba2af8cec90408593ff913f04441847c1b1f230"}]},"branch":"refs/heads/master"},"f5cab7a802f83f1c368955133ed14645b7b15578":{"kind":"REWORK","_number":8,"created":"2019-08-29 13:22:32.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/8"}}},"commit":{"parents":[{"commit":"99484b1a1ae1d99a760ac213c7d319bd407c8f6b","subject":"PowerMax Driver - Miscellaneous improvements to delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99484b1a1ae1d99a760ac213c7d319bd407c8f6b"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-08-29 13:22:01.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5cab7a802f83f1c368955133ed14645b7b15578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5cab7a802f83f1c368955133ed14645b7b15578"}]},"branch":"refs/heads/master"},"e90b921e2249ab171478deabf0d5a9bfe929d5fd":{"kind":"REWORK","_number":9,"created":"2019-08-30 09:12:28.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/9"}}},"commit":{"parents":[{"commit":"5fcd46e872967f947dfe8a10b6ada352659d2b98","subject":"PowerMax Driver - Miscellaneous improvements to delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5fcd46e872967f947dfe8a10b6ada352659d2b98"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-08-30 09:12:14.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e90b921e2249ab171478deabf0d5a9bfe929d5fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e90b921e2249ab171478deabf0d5a9bfe929d5fd"}]},"branch":"refs/heads/master"},"4d968acc947fa5bc3296ebedee6b51b43786c63e":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-08-30 13:52:29.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/35/671235/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/671235/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/671235/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/671235/10"}}},"commit":{"parents":[{"commit":"3d0abc28fb160572962f90f8e35be987f024cb8b","subject":"PowerMax Driver - Miscellaneous improvements to delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d0abc28fb160572962f90f8e35be987f024cb8b"}]}],"author":{"name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","date":"2019-07-17 06:54:31.000000000","tz":60},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2019-08-30 13:47:02.000000000","tz":60},"subject":"PowerMax Driver - SnapVX NoCopy Mode","message":"PowerMax Driver - SnapVX NoCopy Mode\n\nThe PowerMax for Cinder driver now implements noCopy mode\nfor links between SnapVX source and target.  This change\nwill improve space efficiency by using pointers instead of\ncopied tracks when source and target volumes are linked.\n\nChange-Id: Idfc8de789a27f54d9211be3c1ea6778586f85672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d968acc947fa5bc3296ebedee6b51b43786c63e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d968acc947fa5bc3296ebedee6b51b43786c63e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
