)]}'
{"id":"openstack%2Fcinder~671416","triplet_id":"openstack%2Fcinder~stable%2Frocky~Ib53ce55cc62ae28d1d9f0d9c390947cafddcc0ea","project":"openstack/cinder","branch":"stable/rocky","topic":"bug/1826220","hashtags":[],"change_id":"Ib53ce55cc62ae28d1d9f0d9c390947cafddcc0ea","subject":"Let error information more precisely.","status":"ABANDONED","created":"2019-07-18 07:26:41.000000000","updated":"2019-07-19 06:08:40.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"373840716d6c07262b24e8af352dc92ce334bdb1","_number":671416,"virtual_id_number":671416,"owner":{"_account_id":30147,"name":"ye","email":"wmxy123@yeah.net","username":"dakele"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-18 08:58:41.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"date":"2019-07-18 16:31:21.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2019-07-18 08:35:50.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2019-07-18 09:16:10.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"date":"2019-07-18 08:29:25.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"date":"2019-07-18 11:21:47.000000000","_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-18 10:47:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":-2,"date":"2019-07-18 12:28:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-18 07:35:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 07:35:39.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2019-07-18 08:29:25.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-07-18 08:35:50.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-18 08:58:41.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-07-18 09:16:10.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-07-18 10:47:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-18 11:21:47.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-18 12:28:06.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-07-18 16:31:21.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bea65ecc3b3f09ce3ee9a7ef92ec89549294966f","author":{"_account_id":30147,"name":"ye","email":"wmxy123@yeah.net","username":"dakele"},"date":"2019-07-18 07:26:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"132ca207875cb3104606cf7a22ae654253dc344c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2019-07-18 07:35:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nFirstly a direct patch to a stable branch is wrong practice, it should first be proposed to master and then backported reverse chronologically as per releases.\n\nSecondly, the code proposed is already being taken care of in a better/optimal way, so i don\u0027t understand the intention behind proposing this patch.","accounts_in_message":[],"_revision_number":1},{"id":"9a75aad08124afeb0137796387c5bbe8a825d102","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-18 08:29:25.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/671416/1/iscsi/ : FAILURE in 45m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/671416/1/smb/ : FAILURE in 1h 02m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ef539a97e75775e2920b9e95ec98faedd29b6914","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-18 08:35:50.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/16/671416/1/check/tempest-dsvm-full-xenial/b687a47/ : FAILURE in 58m 14s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/16/671416/1/check/devstack-plugin-ceph-dsvm-xenial/870e9ee/ : FAILURE in 1h 08m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5278e8425ae212e949d4c7b9d1f86effcf8ef1ba","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-18 08:58:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/16/671416/1/check/dsvm-tempest-cisco-zonemanager-job/00db579/ : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1b5a3aa8975649a5837cbf39c7e0cc666496ffa7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-18 09:16:10.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/16/671416/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"623114be280b990473f9966313844bdeac280024","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 10:47:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/16/671416/1/check/openstack-tox-pep8/6b8e60a/ : FAILURE in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/16/671416/1/check/openstack-tox-py27/51f60b0/ : FAILURE in 13m 42s\n- openstack-tox-py35 http://logs.openstack.org/16/671416/1/check/openstack-tox-py35/932781f/ : FAILURE in 9m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/671416/1/check/build-openstack-sphinx-docs/35cb5e0/html/ : SUCCESS in 6m 37s\n- tempest-full http://logs.openstack.org/16/671416/1/check/tempest-full/820d10e/ : FAILURE in 1h 14m 53s\n- neutron-grenade http://logs.openstack.org/16/671416/1/check/neutron-grenade/d28ed63/ : SUCCESS in 53m 34s\n- tempest-full-py3 http://logs.openstack.org/16/671416/1/check/tempest-full-py3/f747c97/ : FAILURE in 1h 19m 05s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/16/671416/1/check/cinder-tempest-dsvm-lvm-lio-barbican/46c294d/ : FAILURE in 2h 10m 20s\n- cinder-tox-bandit-baseline http://logs.openstack.org/16/671416/1/check/cinder-tox-bandit-baseline/d8a6da8/ : SUCCESS in 7m 27s (non-voting)\n- nova-multiattach http://logs.openstack.org/16/671416/1/check/nova-multiattach/923e157/ : SUCCESS in 1h 02m 19s\n- openstack-tox-functional-py35 http://logs.openstack.org/16/671416/1/check/openstack-tox-functional-py35/b6d53b7/ : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/671416/1/check/openstack-tox-lower-constraints/211bcfd/ : FAILURE in 10m 53s\n- cinder-tox-functional-py36 http://logs.openstack.org/16/671416/1/check/cinder-tox-functional-py36/d9903af/ : SUCCESS in 5m 49s\n- cinder-tox-py36 http://logs.openstack.org/16/671416/1/check/cinder-tox-py36/32e1239/ : FAILURE in 9m 28s\n- cinder-rally-task http://logs.openstack.org/16/671416/1/check/cinder-rally-task/0c51554/ : SUCCESS in 44m 55s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/16/671416/1/check/openstack-tox-pylint/8f93b4d/ : FAILURE in 42m 49s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/16/671416/1/check/devstack-plugin-ceph-tempest/399b8c5/ : FAILURE in 1h 11m 03s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/16/671416/1/check/openstack-tox-functional/3820f3d/ : SUCCESS in 4m 31s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/16/671416/1/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/fdf28d7/ : SUCCESS in 1h 29m 28s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/16/671416/1/check/legacy-tempest-dsvm-lvm-multibackend/a413224/ : FAILURE in 1h 33m 09s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack http://logs.openstack.org/16/671416/1/check/legacy-tempest-dsvm-full-drbd-devstack/e9669d7/ : FAILURE in 1h 23m 07s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/16/671416/1/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/78d5999/ : FAILURE in 1h 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ec6fb2bb9ee437522831f39d78998bcaff2a4fc","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-07-18 11:21:47.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/16/671416/1/check/Cinder-AS13000/ec96752 : FAILURE in 9m 18s\n- cinder-test-fc http://218.29.111.78:9000/16/671416/1/check/cinder-test-fc/97a03b5 : FAILURE in 31m 03s\n- cinder-test-iscsi http://218.29.111.78:9000/16/671416/1/check/cinder-test-iscsi/b419406 : FAILURE in 1h 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a08dfd5f5e723ff5c14bc27c20ae0b888ecc0556","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-18 12:28:06.000000000","message":"Patch Set 1: Code-Review-2\n\nNeeds to be submitted to master first, it\u0027s also not clear whether this patch is correct.","accounts_in_message":[],"_revision_number":1},{"id":"cbbb6dd3884fe83c53d6a409c809963a23191cef","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-18 16:31:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/16/671416/1/check-cinder/Cinder-ISER-ISCSI/8da5a80 : SUCCESS in 1h 06m 30s\n- Cinder-ISER-LIO http://13.74.249.42/16/671416/1/check-cinder/Cinder-ISER-LIO/c452f94 : SUCCESS in 1h 13m 06s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"bef0d2dcb7c27f00aaadec59d29d6e0f19788652","author":{"_account_id":30147,"name":"ye","email":"wmxy123@yeah.net","username":"dakele"},"date":"2019-07-19 06:08:40.000000000","message":"Abandoned\n\nthanks a lot, abandon this change.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21842741cb41b5ee8482f1bbf28b6d191a969da3","revisions":{"21842741cb41b5ee8482f1bbf28b6d191a969da3":{"kind":"REWORK","_number":1,"created":"2019-07-18 07:26:41.000000000","uploader":{"_account_id":30147,"name":"ye","email":"wmxy123@yeah.net","username":"dakele"},"ref":"refs/changes/16/671416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/671416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/671416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/671416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/671416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/671416/1"}}},"commit":{"parents":[{"commit":"2990af850e54ba746e6ca9620d3e817e7e52e5b1","subject":"Remove Sheepdog tests from zuul config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2990af850e54ba746e6ca9620d3e817e7e52e5b1"}]}],"author":{"name":"dakele","email":"wmxy123@yeah.net","date":"2019-07-18 07:13:23.000000000","tz":480},"committer":{"name":"dakele","email":"wmxy123@yeah.net","date":"2019-07-18 07:13:23.000000000","tz":480},"subject":"Let error information more precisely.","message":"Let error information more precisely.\n\nThis fix check volume status before conditional_update when users trying to delete volume.\n\nChange-Id: Ib53ce55cc62ae28d1d9f0d9c390947cafddcc0ea\nPartial-Bug:1826220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21842741cb41b5ee8482f1bbf28b6d191a969da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21842741cb41b5ee8482f1bbf28b6d191a969da3"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
