)]}'
{"id":"openstack%2Fcinder~673851","triplet_id":"openstack%2Fcinder~stable%2Fstein~Ic2463338b698c5cf805c0ae06d0229f54f64b3fc","project":"openstack/cinder","branch":"stable/stein","hashtags":[],"change_id":"Ic2463338b698c5cf805c0ae06d0229f54f64b3fc","subject":"Prevent double-attachment race in attachment_reserve","status":"MERGED","created":"2019-07-31 15:34:33.000000000","updated":"2019-08-12 22:26:12.000000000","submitted":"2019-08-12 22:26:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673851-1565648772611-25fc65c6","meta_rev_id":"2baa65bc55df53505637c733bde0ffdc495cb5f7","_number":673851,"virtual_id_number":673851,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-07-31 21:09:58.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-07-31 23:55:35.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-31 17:02:06.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-01 14:10:59.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-07-31 16:50:44.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":2,"date":"2019-08-12 22:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-31 16:46:06.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-31 23:26:33.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2019-07-31 17:14:47.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2019-07-31 22:59:41.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-31 22:49:43.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-08-01 00:53:13.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-01 22:23:17.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-12 20:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":2,"date":"2019-08-12 19:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-12 20:29:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 16:46:06.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-31 16:50:44.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-07-31 17:02:06.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-31 17:14:47.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-07-31 21:09:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-07-31 22:49:43.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-07-31 22:59:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-31 23:26:33.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-07-31 23:55:35.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 00:53:13.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 14:10:59.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-08-01 22:23:17.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-12 19:04:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-08-12 20:29:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-12 22:26:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"534aaf758669ede494e4e0797559b6eb4c0674a6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-31 15:34:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cf8c190c9ba3149fd88371f5fb9aa283896029a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-31 15:34:46.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting for master branch merge of\n    https://review.opendev.org/#/c/671370/","accounts_in_message":[],"_revision_number":1},{"id":"8f0237a56f845c971f1c030ae9ca065631160f2b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-31 15:35:53.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/51/673851/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"94ef50a0cd4fa7f3cbad7e8487e59449ad56d955","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-07-31 15:39:09.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0beee9ee851264548b18efc885a87bd14e1f0cb4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-31 16:46:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/51/673851/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/2ca2fb0 : SUCCESS in 53m 01s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/51/673851/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/09d4728 : SUCCESS in 1h 06m 38s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"3e2421fbd20299e012eff382fc2aa384d84ca42c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-07-31 16:50:44.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/673851/2/iscsi/ : FAILURE in 1h 11m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/673851/2/smb/ : FAILURE in 59m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ede60752fa6c4f8a00f4a45412c9ca12da041a6c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-31 17:02:06.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/51/673851/2/check/tempest-dsvm-full-xenial/52f00ce/ : SUCCESS in 1h 21m 32s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/51/673851/2/check/devstack-plugin-ceph-dsvm-xenial/a6f5490/ : SUCCESS in 1h 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e77bccd0c50d2e6080d8e3dc98f19042b4523163","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-07-31 17:14:47.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/51/673851/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"fed3968579596eaaaf8781157a718bfbd0eb8fa9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-07-31 21:09:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/51/673851/2/check/dsvm-tempest-cisco-zonemanager-job/d9335c0/ : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0c9300b4a5fc1378060bb1e891e58e4e85e2dd5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 21:25:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/51/673851/2/check/openstack-tox-lower-constraints/4141364/ : SUCCESS in 10m 51s\n- openstack-tox-pep8 http://logs.openstack.org/51/673851/2/check/openstack-tox-pep8/0f90713/ : SUCCESS in 7m 55s\n- openstack-tox-py27 http://logs.openstack.org/51/673851/2/check/openstack-tox-py27/2d04aa7/ : SUCCESS in 9m 49s\n- openstack-tox-py36 http://logs.openstack.org/51/673851/2/check/openstack-tox-py36/5d2fb11/ : SUCCESS in 9m 12s\n- openstack-tox-docs http://logs.openstack.org/51/673851/2/check/openstack-tox-docs/a640401/html/ : SUCCESS in 8m 45s\n- tempest-full http://logs.openstack.org/51/673851/2/check/tempest-full/0cfbffd/ : SUCCESS in 1h 52m 19s\n- neutron-grenade http://logs.openstack.org/51/673851/2/check/neutron-grenade/c5cbac4/ : SUCCESS in 1h 15m 52s\n- grenade-py3 http://logs.openstack.org/51/673851/2/check/grenade-py3/d5218f6/ : SUCCESS in 1h 18m 58s\n- tempest-full-py3 http://logs.openstack.org/51/673851/2/check/tempest-full-py3/17b1893/ : SUCCESS in 1h 42m 18s\n- cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/51/673851/2/check/cinder-tempest-dsvm-lvm-lio-barbican/07f85a0/ : SUCCESS in 1h 58m 13s\n- cinder-tox-bandit-baseline http://logs.openstack.org/51/673851/2/check/cinder-tox-bandit-baseline/e091a7e/ : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-functional-py36 http://logs.openstack.org/51/673851/2/check/openstack-tox-functional-py36/4077daf/ : SUCCESS in 8m 06s\n- cinder-rally-task http://logs.openstack.org/51/673851/2/check/cinder-rally-task/9fc906c/ : SUCCESS in 44m 18s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/51/673851/2/check/openstack-tox-pylint/08dc40f/ : SUCCESS in 5m 54s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/51/673851/2/check/devstack-plugin-ceph-tempest/b203906/ : SUCCESS in 1h 31m 21s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/51/673851/2/check/openstack-tox-functional/9c08b7a/ : SUCCESS in 4m 22s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak http://logs.openstack.org/51/673851/2/check/legacy-grenade-dsvm-cinder-mn-sub-volbak/8edeb70/ : SUCCESS in 1h 07m 25s\n- legacy-tempest-dsvm-lvm-multibackend http://logs.openstack.org/51/673851/2/check/legacy-tempest-dsvm-lvm-multibackend/6255dab/ : SUCCESS in 1h 45m 21s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs http://logs.openstack.org/51/673851/2/check/legacy-tempest-dsvm-full-devstack-plugin-nfs/24102d7/ : SUCCESS in 1h 38m 42s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/51/673851/2/check/tempest-slow-py3/fabaa63/ : SUCCESS in 2h 34m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ace60d6f9df06dc5dd7f5a1ab2c7ea89da38076a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-07-31 22:49:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/51/673851/2/upstream-solidfire-check/cinder-solidfire-iSCSI/f634336/ : SUCCESS in 1h 18m 35s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"a95f84599957e662bc94ad96f3806a293794585f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-31 22:59:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.161.150/logs/51/673851/2/upstream-check/cinder-cDOT-iSCSI/b0f9564/ : SUCCESS in 57m 43s\n- cinder-cDOT-NFS http://13.56.161.150/logs/51/673851/2/upstream-check/cinder-cDOT-NFS/5120140/ : SUCCESS in 1h 03m 00s\n- cinder-cDOT-FCP http://13.56.161.150/logs/51/673851/2/upstream-check/cinder-cDOT-FCP/11887f9/ : SUCCESS in 1h 29m 01s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"66d9285004496f14b86e366f130c158cb2e9f081","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-07-31 23:26:33.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/51/673851/2/check/kaminario-dsvm-tempest-full-iscsi/0446ccf : SUCCESS in 2h 00m 18s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/51/673851/2/check/kaminario-dsvm-tempest-full-FC/b8faeec : FAILURE in 25m 53s","accounts_in_message":[],"_revision_number":2},{"id":"39770a4c8e463d5767e88532ef3528acc28a55fb","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-31 23:55:35.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/51/673851/2/check-cinder/Cinder-ISER-ISCSI/aaccfb5 : FAILURE in 1h 16m 46s\n- Cinder-ISER-LIO http://13.74.249.42/51/673851/2/check-cinder/Cinder-ISER-LIO/560ab16 : SUCCESS in 1h 10m 35s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"009a346700101d58dcff0ac17c0c272ba724ef65","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-08-01 00:53:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/51/673851/2/check/dsvm-tempest-zadara-driver/5768eb6 : SUCCESS in 1h 47m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"953fd6742aee6eb9197a27d0310a92893288d075","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-01 14:10:59.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/51/673851/2/check/dsvm-tempest-infinibox-fc/4f1c11b : SUCCESS in 58m 14s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/51/673851/2/check/dsvm-tempest-infinibox-iscsi/fcf3126 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"533aaf37090989cc83e103dcd279b192116d9518","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 22:23:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/51/673851/2/check/NEC-ISCSI/79a2182 : SUCCESS in 2h 07m 49s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/51/673851/2/check/NEC-FC/d90053a : SUCCESS in 1h 34m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0731aa7eea30b5eebde019e4a1fcd54be79771ca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-12 19:04:16.000000000","message":"Patch Set 2: Code-Review+2\n\nThis is an appropriate backport.","accounts_in_message":[],"_revision_number":2},{"id":"334405205ea49262125fcc43c065f43a7578543c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-12 20:29:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"463b71801d284f43025cbe3a00a493f16f93bce4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:29:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b6b3df177cfab6f2781184e576f9adf5183278d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:26:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/477e8f964b2f49c582d79a20c038d6b7 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de9ce187f31748d5a9ba5089d633fcfa : SUCCESS in 7m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04fb2e1540454ae6bc5b19070c679d89 : SUCCESS in 11m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23cfd6b10a0f4f6e9799efb6afb95910 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/429ad338404f4e82823cbb1b04cc5778 : SUCCESS in 8m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/09372c9bb1464816810a1a11de937b0c : SUCCESS in 1h 53m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7710dfadcbca481ab3522735a31e068e : SUCCESS in 57m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/95fe9f41a31e4c63b3af32983745be8d : SUCCESS in 1h 09m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5df6345da78741fa89fc3f9c1fca54a6 : SUCCESS in 1h 28m 22s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eb7b7ef11a9d4090856fd63c5494fb14 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2baa65bc55df53505637c733bde0ffdc495cb5f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:26:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eadff0df7efbe662bf107f77e1b2ac96c912e4e0","revisions":{"a672893666bd3a5e84b0cbbc196169d4c29462d6":{"kind":"REWORK","_number":1,"created":"2019-07-31 15:34:33.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/673851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/673851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/673851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/673851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/673851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/673851/1"}}},"commit":{"parents":[{"commit":"0996f0ac76188c324831f19089bdd87b9c364cb6","subject":"Merge \"Fix assertion methods in a unit test\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0996f0ac76188c324831f19089bdd87b9c364cb6"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-07-16 17:28:10.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-07-31 15:34:09.000000000","tz":-240},"subject":"Prevent double-attachment race in attachment_reserve","message":"Prevent double-attachment race in attachment_reserve\n\nIf multiple attachments are requested simultaneously,\na volume can be attached to an instance more than once.\n\nThis fixes this problem by detecting that a race has\noccurred in _attachment_reserve, and backing out one of\nthe invalid attachment records.\n\nThis means that if the attachment API is called\nrepeatedly and quickly for the same volume, some requests\nmay fail, but this is much better than incorrectly\ncreating multiple attachments.\n\nCloses-Bug: #1833736\n\nChange-Id: Ic2463338b698c5cf805c0ae06d0229f54f64b3fc\n(cherry picked from commit 7f3a77b66f9aa1b185310c8cbe33b358da6cc39a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a672893666bd3a5e84b0cbbc196169d4c29462d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a672893666bd3a5e84b0cbbc196169d4c29462d6"}]},"branch":"refs/heads/stable/stein"},"eadff0df7efbe662bf107f77e1b2ac96c912e4e0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-31 15:39:09.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/673851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/673851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/673851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/673851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/673851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/673851/2"}}},"commit":{"parents":[{"commit":"0996f0ac76188c324831f19089bdd87b9c364cb6","subject":"Merge \"Fix assertion methods in a unit test\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0996f0ac76188c324831f19089bdd87b9c364cb6"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-07-16 17:28:10.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-07-31 15:38:37.000000000","tz":-240},"subject":"Prevent double-attachment race in attachment_reserve","message":"Prevent double-attachment race in attachment_reserve\n\nIf multiple attachments are requested simultaneously,\na volume can be attached to an instance more than once.\n\nThis fixes this problem by detecting that a race has\noccurred in _attachment_reserve, and backing out one of\nthe invalid attachment records.\n\nThis means that if the attachment API is called\nrepeatedly and quickly for the same volume, some requests\nmay fail, but this is much better than incorrectly\ncreating multiple attachments.\n\nCloses-Bug: #1833736\n\nChange-Id: Ic2463338b698c5cf805c0ae06d0229f54f64b3fc\n(cherry picked from commit 7f3a77b66f9aa1b185310c8cbe33b358da6cc39a)\nDepends-On: https://review.opendev.org/671370\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eadff0df7efbe662bf107f77e1b2ac96c912e4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eadff0df7efbe662bf107f77e1b2ac96c912e4e0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
