)]}'
{"id":"openstack%2Fcinder~675727","triplet_id":"openstack%2Fcinder~master~If86dd619402495d9d4470b14cb270fcf53db6794","project":"openstack/cinder","branch":"master","topic":"internal-links","hashtags":[],"change_id":"If86dd619402495d9d4470b14cb270fcf53db6794","subject":"Update docs building","status":"MERGED","created":"2019-08-10 08:24:09.000000000","updated":"2019-08-14 07:58:36.000000000","submitted":"2019-08-13 00:52:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675727-1565657541131-7bf71c1f","meta_rev_id":"789080aac4b27d23b988f834775068960d5a3561","_number":675727,"virtual_id_number":675727,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-10 10:35:22.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 16:10:28.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2019-08-10 20:20:54.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 19:00:43.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 11:10:33.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-12 02:21:46.000000000","_account_id":21767,"name":"Srinivas","email":"srinivasd@biarca.com","username":"Srinivas"},{"value":0,"date":"2019-08-10 11:12:03.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-11 23:36:40.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-13 00:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 15:58:47.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2019-08-10 10:59:17.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 13:53:39.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-08-12 21:57:02.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"date":"2019-08-12 04:01:36.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2019-08-10 14:56:26.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2019-08-10 11:41:00.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 11:43:50.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 11:18:33.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-08-14 07:58:36.000000000","post_submit":true,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-08-10 12:40:49.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 10:50:24.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2019-08-11 05:59:16.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2019-08-10 11:58:13.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 14:00:05.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 17:08:09.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-10 19:13:46.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2019-08-11 05:53:36.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-10 17:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21767,"name":"Srinivas","email":"srinivasd@biarca.com","username":"Srinivas"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":1,"date":"2019-08-12 02:54:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-08-12 21:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21767,"name":"Srinivas","email":"srinivasd@biarca.com","username":"Srinivas"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-08-12 21:32:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21767,"name":"Srinivas","email":"srinivasd@biarca.com","username":"Srinivas"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21767,"name":"Srinivas","email":"srinivasd@biarca.com","username":"Srinivas"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-10 10:35:22.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 10:50:24.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-08-10 10:59:17.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 11:10:33.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 11:12:03.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 11:18:33.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-08-10 11:41:00.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 11:43:50.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 11:58:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 12:40:49.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 13:53:39.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2019-08-10 14:00:05.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 14:56:26.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-08-10 15:58:47.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-08-10 16:10:28.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2019-08-10 17:08:09.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 17:38:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-10 19:00:43.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 19:13:46.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-08-10 20:20:54.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-11 05:53:36.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-11 05:59:16.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2019-08-11 23:36:40.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-12 02:21:46.000000000","updated_by":{"_account_id":21767,"name":"Srinivas","email":"srinivasd@biarca.com","username":"Srinivas"},"reviewer":{"_account_id":21767,"name":"Srinivas","email":"srinivasd@biarca.com","username":"Srinivas"},"state":"REVIEWER"},{"updated":"2019-08-12 02:54:31.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-08-12 04:01:36.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-08-12 21:32:42.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-08-12 21:57:02.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2019-08-13 00:52:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 07:58:36.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"}],"messages":[{"id":"477e7032367472b01a2ec733aabba4d4e08c22aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-10 08:24:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f96fd8f08d285866e245a5cf7350787b8fe90db","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-10 08:26:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c53029899924e59284f339fc09771dba07b7be3b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-10 08:46:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fd52e2840fe93af8d59ca4ab86e779f32b102924","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-10 08:50:19.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"7f2c47a1856007aea20b23d741e6efc298ab1dbb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-10 09:00:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/27/675727/4/check/fujitsu-eternus-dx-iscsi/4944dce : FAILURE in 9m 13s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/27/675727/4/check/fujitsu-eternus-dx-fc/2a99f8e : FAILURE in 1m 00s","accounts_in_message":[],"_revision_number":4},{"id":"ea1e999c0e4ff8aba9b0e0684ad2d48884eda32c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 09:33:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/27/675727/4/check/build-openstack-api-ref/5d9e215/html/ : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://logs.opendev.org/27/675727/4/check/openstack-tox-lower-constraints/86f7dfd/ : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://logs.opendev.org/27/675727/4/check/openstack-tox-pep8/649ef45/ : SUCCESS in 7m 34s\n- openstack-tox-py27 https://logs.opendev.org/27/675727/4/check/openstack-tox-py27/f9130b6/ : SUCCESS in 10m 02s\n- openstack-tox-py36 https://logs.opendev.org/27/675727/4/check/openstack-tox-py36/d3f415a/ : SUCCESS in 9m 40s\n- openstack-tox-py37 https://logs.opendev.org/27/675727/4/check/openstack-tox-py37/b0681ff/ : SUCCESS in 10m 16s\n- openstack-tox-docs https://logs.opendev.org/27/675727/4/check/openstack-tox-docs/09538f1/html/ : SUCCESS in 6m 42s\n- requirements-check https://logs.opendev.org/27/675727/4/check/requirements-check/d09cf24/ : FAILURE in 2m 42s\n- build-openstack-releasenotes https://logs.opendev.org/27/675727/4/check/build-openstack-releasenotes/1aa7349/html/ : SUCCESS in 12m 25s\n- openstack-tox-functional-py36 https://logs.opendev.org/27/675727/4/check/openstack-tox-functional-py36/89ea563/ : SUCCESS in 4m 47s\n- cinder-rally-task https://logs.opendev.org/27/675727/4/check/cinder-rally-task/c49b749/ : SUCCESS in 41m 27s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/27/675727/4/check/openstack-tox-pylint/2454ab1/ : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/27/675727/4/check/openstack-tox-functional/7f793cb/ : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a3b18a6bc2664047b2cd9c2a15c35018b84b562c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-10 09:39:16.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"213aacc9f942781bf65bb3d0867d6c799b9c7ef3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-10 09:42:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"28356f107a12f56a7a078fb46b1fbe29c1848641","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 10:21:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://logs.opendev.org/27/675727/6/check/build-openstack-api-ref/de28af1/html/ : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://logs.opendev.org/27/675727/6/check/openstack-tox-lower-constraints/bf23fba/ : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://logs.opendev.org/27/675727/6/check/openstack-tox-pep8/649ecd8/ : SUCCESS in 7m 47s\n- openstack-tox-py27 https://logs.opendev.org/27/675727/6/check/openstack-tox-py27/e08d76e/ : SUCCESS in 8m 34s\n- openstack-tox-py36 https://logs.opendev.org/27/675727/6/check/openstack-tox-py36/81ae24f/ : SUCCESS in 8m 16s\n- openstack-tox-py37 https://logs.opendev.org/27/675727/6/check/openstack-tox-py37/03f493b/ : SUCCESS in 7m 50s\n- openstack-tox-docs https://logs.opendev.org/27/675727/6/check/openstack-tox-docs/4e1accc/html/ : SUCCESS in 7m 02s\n- requirements-check https://logs.opendev.org/27/675727/6/check/requirements-check/dfacba4/ : SUCCESS in 2m 27s\n- build-openstack-releasenotes https://logs.opendev.org/27/675727/6/check/build-openstack-releasenotes/7dcfd74/html/ : SUCCESS in 13m 32s\n- openstack-tox-functional-py36 https://logs.opendev.org/27/675727/6/check/openstack-tox-functional-py36/fee9d5f/ : SUCCESS in 4m 24s\n- cinder-rally-task https://logs.opendev.org/27/675727/6/check/cinder-rally-task/322ba30/ : SUCCESS in 37m 48s (non-voting)\n- openstack-tox-pylint https://logs.opendev.org/27/675727/6/check/openstack-tox-pylint/6842d0e/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/27/675727/6/check/openstack-tox-functional/760337d/ : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a070e56c746eb568e1de5ce4f7b0550a815ce540","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-10 10:35:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/27/675727/6/check/dsvm-tempest-kvm/863853d : FAILURE in 35m 30s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/27/675727/6/check/dsvm-tempest-exe/eb47e17 : FAILURE in 35m 22s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"999717e55dcf6535278b4e3aae066ba4dad15470","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-08-10 10:50:24.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/675727/6/iscsi/ : FAILURE in 1h 07m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/675727/6/smb/ : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":6},{"id":"0b36d4e4542fd86ec93294db5a48113a56ff44c2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-10 10:59:17.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/27/675727/6/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/a0145f4 : FAILURE in 58m 39s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/27/675727/6/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/da832da : SUCCESS in 1h 16m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"86f178b99816216e9eecc001745acef2180b3922","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-10 11:10:33.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/27/675727/6/check/tempest-dsvm-full-xenial/28b5b08/ : SUCCESS in 1h 21m 43s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/27/675727/6/check/devstack-plugin-ceph-dsvm-xenial/d48da9f/ : SUCCESS in 1h 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"63f9fb2c7ecae84135546aa025b2974d1bdb7266","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2019-08-10 11:12:03.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/27/675727/6/check/fujitsu-eternus-dx-iscsi/f4cac9c : SUCCESS in 1h 29m 16s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/27/675727/6/check/fujitsu-eternus-dx-fc/6b9478a : FAILURE in 2m 24s","accounts_in_message":[],"_revision_number":6},{"id":"0961976076f7b4f62566b27a7a803c07d78c35a9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-10 11:18:33.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/27/675727/6/upstream-solidfire-check/cinder-solidfire-iSCSI/97b2411/ : SUCCESS in 55m 24s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"4fd0c792aa60347578631e261f079dbd7137b291","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-10 11:23:05.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/27/675727/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c37c8a4ed869ca10856a1777cde6289448feb412","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-10 11:33:55.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/27/675727/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"24d3224c85cb5bf3e891d5291c748792ea82cb2a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-08-10 11:41:00.000000000","message":"Patch Set 6:\n\n* nexenta-dsvm-volume-ns5_iscsi http://104.254.242.106/refs-changes-27-675727-6/ns5_iscsi : FAILURE in 51m 14s\n* nexenta-dsvm-volume-ns5_nfs http://104.254.242.106/refs-changes-27-675727-6/ns5_nfs : FAILURE in 54m 24s","accounts_in_message":[],"_revision_number":6},{"id":"29fb82976f7261b330e8c29214b304748fdce61f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-10 11:43:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.161.150/logs/27/675727/6/upstream-check/cinder-cDOT-iSCSI/38259b4/ : SUCCESS in 1h 21m 23s\n- cinder-cDOT-NFS http://13.56.161.150/logs/27/675727/6/upstream-check/cinder-cDOT-NFS/6942925/ : SUCCESS in 48m 28s\n- cinder-cDOT-FCP http://13.56.161.150/logs/27/675727/6/upstream-check/cinder-cDOT-FCP/f8f4ab9/ : SUCCESS in 1h 01m 59s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"b9aaaa87e73549d04d1defb7f8410d27309019f8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-10 11:58:13.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/27/675727/6/check-not-vote/ext-cinder-zuul/a249db6 : FAILURE in 1m 49s","accounts_in_message":[],"_revision_number":6},{"id":"d639c9ee8edce451374f21d16f6a78b5256292c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-10 12:40:49.000000000","message":"Patch Set 6:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/27/675727/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2e0e913 : SUCCESS in 2h 15m 46s\n- 3par-fc-driver-master-client-pip-ssa02-dsvm http://54.201.44.218/27/675727/6/check/3par-fc-driver-master-client-pip-ssa02-dsvm/04b49d3 : SUCCESS in 2h 16m 32s\n- lefthand-iscsi-driver-master-client-pip-vsa126-dsvm http://54.201.44.218/27/675727/6/check/lefthand-iscsi-driver-master-client-pip-vsa126-dsvm/ac205ab : FAILURE in 1h 30m 25s","accounts_in_message":[],"_revision_number":6},{"id":"388ab09c0a7b70d73c047814870c884456b50cec","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-10 13:02:32.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/27/675727/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ba8ec25856115d397b2945e4f66c8d58bfe112ef","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-10 13:16:43.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/27/675727/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"5d5b11c1a52ea73c9f6ea995309059cc2d482d1f","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-08-10 13:53:39.000000000","message":"Patch Set 6:\n\nBuild Successful. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/27/2504 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"10837c6528ddcb1851047e41cd4350848ba7bf97","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-10 14:00:05.000000000","message":"Patch Set 6:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-27-675727-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"534aae7cd51a12fa60df65d63dcc24220f60f9e0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-10 14:47:50.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/27/675727/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"52ec06d5cf15dc2801bb7be1ce403a808efedb86","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-08-10 14:56:26.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/27/675727/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"ee190bd0652f7459a55fd9147a5f52dbefb39329","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-10 15:58:47.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/27/675727/6/check/DellEMC_SC_FC/46a1e4b/DellEMC_SC_FC/433 : SUCCESS in 1h 07m 57s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/27/675727/6/check/DellEMC_SC_iSCSI/02312ff/DellEMC_SC_iSCSI/414 : SUCCESS in 1h 37m 13s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"52c9dd759c79ca8fd9debacd2b92e290a726140d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2019-08-10 16:10:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/27/675727/6/check/dsvm-tempest-cisco-zonemanager-job/dfcd549/ : FAILURE in 1h 06m 41s","accounts_in_message":[],"_revision_number":6},{"id":"df8320c84f7214b50ead0747dbe19a4d2433f57b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-10 17:08:09.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/27/675727/6/check/DellEMC_VNX_iSCSI/2f59d85/DellEMC_VNX_iSCSI/566 : SUCCESS in 1h 38m 20s\n- DellEMC_VNX_FC http://publiclogs.emc.com/27/675727/6/check/DellEMC_VNX_FC/abded9c/DellEMC_VNX_FC/577 : FAILURE in 17m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"213b42f0b5dbfd8adcb2b161f66145a846d0764a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-10 17:38:24.000000000","message":"Patch Set 6: Code-Review+2\n\nI do kind of prefer having the doc8 job as part of the docs target to make sure it captures doc related issues that don\u0027t actually generate errors or warnings in the build, but better to stay consistent across projects.","accounts_in_message":[],"_revision_number":6},{"id":"2580251f589f99be4e75c38e82ee6b53a2a88d58","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-10 18:00:22.000000000","message":"Patch Set 6:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/27/675727/6/check/kaminario-dsvm-tempest-full-iscsi/0a2a7c8 : FAILURE in 2h 58m 20s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/27/675727/6/check/kaminario-dsvm-tempest-full-FC/944045d : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":6},{"id":"5d03e7c342f8a7fe41567f9b955cc21d6eccc55a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-10 19:00:43.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/27/675727/6/check/DellEMC_VxFlexOS/6007586/DellEMC_VxFlexOS/554 : SUCCESS in 1h 52m 11s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"1e30843dd48e41ea12ff21ec1d46dd54beda785c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-10 19:13:46.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/27/675727/6/check/DellEMC_Unity_FC/cacfb4b/DellEMC_Unity_FC/989 : SUCCESS in 1h 24m 23s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/27/675727/6/check/DellEMC_Unity_iSCSI/8ff5fec/DellEMC_Unity_iSCSI/944 : SUCCESS in 1h 31m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"bd48a335e3387d6154acea3cf2afc0d5999bdfff","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-10 20:20:54.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/27/675727/6/check/DellEMC_XtremIO_FC/adc648b/DellEMC_XtremIO_FC/1116 : SUCCESS in 1h 30m 31s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/27/675727/6/check/DellEMC_XtremIO_iSCSI/dba2a1f/DellEMC_XtremIO_iSCSI/1038 : SUCCESS in 1h 27m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"a7008803a2eee6d2e063eee01f1654041054772b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-08-11 05:53:36.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/27/675727/6/check/NEC-ISCSI/6939f9f : SUCCESS in 1h 42m 08s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/27/675727/6/check/NEC-FC/b0bb341 : SUCCESS in 2h 06m 49s","accounts_in_message":[],"_revision_number":6},{"id":"8caa5c69836163b995cfec27ebe9f7450e3cf13e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2019-08-11 05:59:16.000000000","message":"Patch Set 6:\n\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/675727-6/HPMSA-FC : FAILURE in 50:43.71\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/675727-6/HPMSA-iSCSI : SUCCESS in 44:28.52","accounts_in_message":[],"_revision_number":6},{"id":"448534bd7da52b82d9147896c27525e74d4c531c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-11 14:21:53.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/675727/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9e5ce523c56455c61b18bb7d145980e8f6817a7d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-11 15:22:49.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/675727/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3985926b3823eb274c38882d6f0b1fb21bf78ab8","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-11 17:33:16.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/675727/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"f4998fe3dfdb2701187690ae79fe01179adc7918","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-11 19:36:43.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/675727/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"155c41c271887eb5c43ef8bbb8a188eededfa3cf","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-11 23:26:33.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/675727/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"68a250c9ca6394c5e2347f24f18cbe1b36ac2652","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2019-08-11 23:36:40.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/675727/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"fef1bd3236598a7fb0c48feb8fabf72d37ce956b","author":{"_account_id":21767,"name":"Srinivas","email":"srinivasd@biarca.com","username":"Srinivas"},"date":"2019-08-12 02:21:46.000000000","message":"Patch Set 6:\n\nrun-Kaminario K2 CI","accounts_in_message":[],"_revision_number":6},{"id":"15675b4147d993b5cf144cd324244143e3ab1db6","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-08-12 02:54:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d79dec7fdd60707b829e31115337558488e43ec1","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-12 04:01:36.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/27/675727/6/check/kaminario-dsvm-tempest-full-iscsi/4f01a6d : SUCCESS in 1h 39m 22s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/27/675727/6/check/kaminario-dsvm-tempest-full-FC/039e6fe : SUCCESS in 1h 34m 17s","accounts_in_message":[],"_revision_number":6},{"id":"9b26561d3f815643813c3ab3c0613b5e3c3fb8a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-12 13:28:44.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/409/675727/1/check/infortrend-cinder-volume-fc/9d54a0f97040265b9d02eee47b9246a0312d22bb : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"aa1e55c4a9aee9e3266eaa2a0642111a4e17f816","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-12 14:28:02.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/410/675727/2/check/infortrend-cinder-volume-fc/97dd9d1ba2d3a384e6146812650e79e256591140 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"6a29ca1cdcf20d6ec20f1248cd0ffd30a1a1cbbf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-12 15:26:43.000000000","message":"Patch Set 3:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/411/675727/3/check/infortrend-cinder-volume-fc/6bb9de6f51adde814d2a7556094808e090061ced : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"6b64fe014431f46feab970da8d8abf5a0679f70a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-12 17:18:53.000000000","message":"Patch Set 4:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/413/675727/4/check/infortrend-cinder-volume-fc/d510dd0444176959882efb84c2a7dce20e34b695 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"8d91f433b1543a8cdc1770115e93bc91f1f6fa45","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-12 18:14:53.000000000","message":"Patch Set 4:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/414/675727/4/check/infortrend-cinder-volume-fc/d510dd0444176959882efb84c2a7dce20e34b695 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"01e65eb12e12d728486f39e9941a29dccb07d7d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-12 19:10:33.000000000","message":"Patch Set 5:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/415/675727/5/check/infortrend-cinder-volume-fc/33efadb678c7c3d0b6a7b66fc9dab8284667b919 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"cb7dfaf123d58acf09e61ea7bf552cd62b995553","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-12 20:06:34.000000000","message":"Patch Set 6:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/416/675727/6/check/infortrend-cinder-volume-fc/af04cc3392f7361d9537c49510447b63e8802b31 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"c3a0626d7b47afabb6c506f4016155be748efd02","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-08-12 21:32:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ccb632985d78c470875fa8f95b955372ec346274","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 21:33:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"230d5cbec8174a5734fc468d036833345866741c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-08-12 21:57:02.000000000","message":"Patch Set 6:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/418/675727/6/check/infortrend-cinder-volume-fc/af04cc3392f7361d9537c49510447b63e8802b31 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"a15c2ea8709576ea81fe40eb0bafc364b7586b4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 00:52:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9c8cd07aca6d4b7da1cdbd97f5508a5c : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c25c3b78ed1a4237a58837aee27b2e40 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2d54600249452faec659a82f6ccde6 : SUCCESS in 8m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ecf88f74d6b9443d8f623f0ab67e7091 : SUCCESS in 10m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa6a3a2c0bed4915acb0399a645cb5a2 : SUCCESS in 9m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd24855cd951418c838912c77a8f1aae : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aafd1b127deb4f0d9e01a2bda5ba8713 : SUCCESS in 8m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3efe8c3c67f744f7b1dc4d61fc5a9331 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/866a7014f44e4c0fbc487461bcd73912 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":6},{"id":"5d119f16660e62f23a73f9c9b0d07de96fcbd82d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 00:52:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"f9c056b791cd292e7563a5daefc9c170a1524509","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2019-08-13 20:39:03.000000000","message":"Patch Set 6:\n\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/675727-6/Lenovo-FC : SUCCESS in 36:26.39\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/675727-6/Lenovo-iSCSI : FAILURE in 2:22:41","accounts_in_message":[],"_revision_number":6},{"id":"d805caf3285c62bb6d4e768cb5a50aef77f3d838","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-08-14 07:58:36.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/27/675727/6/check/dsvm-tempest-infinibox-fc/a8e75eb : FAILURE in 19m 06s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/27/675727/6/check/dsvm-tempest-infinibox-iscsi/dba2a8a : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"af04cc3392f7361d9537c49510447b63e8802b31","revisions":{"9d54a0f97040265b9d02eee47b9246a0312d22bb":{"kind":"REWORK","_number":1,"created":"2019-08-10 08:24:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/27/675727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/675727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/675727/1"}}},"commit":{"parents":[{"commit":"6cc03512e64e5ac6e460d0afd818a1ccd9bc4f74","subject":"Docs: Fix broken link and make more robust","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cc03512e64e5ac6e460d0afd818a1ccd9bc4f74"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:11:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:21:49.000000000","tz":120},"subject":"Update docs building","message":"Update docs building\n\nCleanup doc/requirements to just use what is needed for doc building.\nMove reno to doc/requirements, add doc8 to test-requirements.\n\nupdate tox.ini:\n* doc8 should be in pep8 as linter and not docs according to PTI [1]\n* Only build main doc in docs environment, CI uses this and we waste\n  just time building api-ref as well, there\u0027s a separate environment for\n  this. This follows PTI [1]\n* Do not include requirements file for docs building, it\u0027s not needed\n  with apidoc.\n\nCleanup doc/source/conf.py, with the switch to api-doc a lot of settings\nare not needed anymore.\n\nUpdate openstackdocstheme to 1.20.0 and which allows to remove obsolete\nsetting of html_last_updated_fmt and project from conf.py.\n\n[1] https://governance.openstack.org/tc/reference/project-testing-interface.html#documentation\n\nChange-Id: If86dd619402495d9d4470b14cb270fcf53db6794\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d54a0f97040265b9d02eee47b9246a0312d22bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d54a0f97040265b9d02eee47b9246a0312d22bb"}]},"branch":"refs/heads/master"},"97dd9d1ba2d3a384e6146812650e79e256591140":{"kind":"REWORK","_number":2,"created":"2019-08-10 08:26:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/27/675727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/675727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/675727/2"}}},"commit":{"parents":[{"commit":"6cc03512e64e5ac6e460d0afd818a1ccd9bc4f74","subject":"Docs: Fix broken link and make more robust","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cc03512e64e5ac6e460d0afd818a1ccd9bc4f74"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:11:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:26:47.000000000","tz":120},"subject":"Update docs building","message":"Update docs building\n\nCleanup doc/requirements to just use what is needed for doc building.\nMove reno to doc/requirements, add doc8 to test-requirements.\n\nupdate tox.ini:\n* doc8 should be in pep8 as linter and not docs according to PTI [1]\n* Only build main doc in docs environment, CI uses this and we waste\n  just time building api-ref as well, there\u0027s a separate environment for\n  this. This follows PTI [1]\n* Do not include requirements file for docs building, it\u0027s not needed\n  with apidoc.\n\nCleanup doc/source/conf.py, with the switch to api-doc a lot of settings\nare not needed anymore.\n\nUpdate openstackdocstheme to 1.20.0 and which allows to remove obsolete\nsetting of html_last_updated_fmt, project, latex_elements from conf.py.\n\n[1] https://governance.openstack.org/tc/reference/project-testing-interface.html#documentation\n\nChange-Id: If86dd619402495d9d4470b14cb270fcf53db6794\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97dd9d1ba2d3a384e6146812650e79e256591140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97dd9d1ba2d3a384e6146812650e79e256591140"}]},"branch":"refs/heads/master"},"6bb9de6f51adde814d2a7556094808e090061ced":{"kind":"REWORK","_number":3,"created":"2019-08-10 08:46:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/27/675727/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/675727/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/675727/3"}}},"commit":{"parents":[{"commit":"6cc03512e64e5ac6e460d0afd818a1ccd9bc4f74","subject":"Docs: Fix broken link and make more robust","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cc03512e64e5ac6e460d0afd818a1ccd9bc4f74"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:11:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:45:52.000000000","tz":120},"subject":"Update docs building","message":"Update docs building\n\nCleanup doc/requirements to just use what is needed for doc building.\nMove reno to doc/requirements, add doc8 to test-requirements.\n\nupdate tox.ini:\n* doc8 should be in pep8 as linter and not docs according to PTI [1]\n* Only build main doc in docs environment, CI uses this and we waste\n  just time building api-ref as well, there\u0027s a separate environment for\n  this. This follows PTI [1]\n* Do not include requirements file for docs building, it\u0027s not needed\n  with apidoc.\n* Use common deps for all docs build environments.\n\nCleanup doc/source/conf.py, with the switch to api-doc a lot of settings\nare not needed anymore, also the eventlet bug is fixed.\n\nUpdate openstackdocstheme to 1.20.0 and which allows to remove obsolete\nsetting of html_last_updated_fmt, project, latex_elements from conf.py.\n\n[1] https://governance.openstack.org/tc/reference/project-testing-interface.html#documentation\n\nChange-Id: If86dd619402495d9d4470b14cb270fcf53db6794\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bb9de6f51adde814d2a7556094808e090061ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bb9de6f51adde814d2a7556094808e090061ced"}]},"branch":"refs/heads/master"},"d510dd0444176959882efb84c2a7dce20e34b695":{"kind":"NO_CHANGE","_number":4,"created":"2019-08-10 08:50:19.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/27/675727/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/675727/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/675727/4"}}},"commit":{"parents":[{"commit":"99ec31f53f5fc4cf62c338a4b2b28b884c2ab623","subject":"Docs: Make links more robust","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99ec31f53f5fc4cf62c338a4b2b28b884c2ab623"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:11:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:49:57.000000000","tz":120},"subject":"Update docs building","message":"Update docs building\n\nCleanup doc/requirements to just use what is needed for doc building.\nMove reno to doc/requirements, add doc8 to test-requirements.\n\nupdate tox.ini:\n* doc8 should be in pep8 as linter and not docs according to PTI [1]\n* Only build main doc in docs environment, CI uses this and we waste\n  just time building api-ref as well, there\u0027s a separate environment for\n  this. This follows PTI [1]\n* Do not include requirements file for docs building, it\u0027s not needed\n  with apidoc.\n* Use common deps for all docs build environments.\n\nCleanup doc/source/conf.py, with the switch to api-doc a lot of settings\nare not needed anymore, also the eventlet bug is fixed.\n\nUpdate openstackdocstheme to 1.20.0 and which allows to remove obsolete\nsetting of html_last_updated_fmt, project, latex_elements from conf.py.\n\n[1] https://governance.openstack.org/tc/reference/project-testing-interface.html#documentation\n\nChange-Id: If86dd619402495d9d4470b14cb270fcf53db6794\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d510dd0444176959882efb84c2a7dce20e34b695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d510dd0444176959882efb84c2a7dce20e34b695"}]},"branch":"refs/heads/master"},"33efadb678c7c3d0b6a7b66fc9dab8284667b919":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-10 09:39:16.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/27/675727/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/675727/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/675727/5"}}},"commit":{"parents":[{"commit":"99ec31f53f5fc4cf62c338a4b2b28b884c2ab623","subject":"Docs: Make links more robust","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99ec31f53f5fc4cf62c338a4b2b28b884c2ab623"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:11:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 09:38:46.000000000","tz":120},"subject":"Update docs building","message":"Update docs building\n\nCleanup doc/requirements to just use what is needed for doc building.\nMove reno to doc/requirements, add doc8 to test-requirements.\n\nupdate tox.ini:\n* doc8 should be in pep8 as linter and not docs according to PTI [1]\n* Only build main doc in docs environment, CI uses this and we waste\n  just time building api-ref as well, there\u0027s a separate environment for\n  this. This follows PTI [1]\n* Do not include requirements file for docs building, it\u0027s not needed\n  with apidoc.\n* Use common deps for all docs build environments.\n\nCleanup doc/source/conf.py, with the switch to api-doc a lot of settings\nare not needed anymore, also the eventlet bug is fixed.\n\nUpdate openstackdocstheme to 1.20.0 and which allows to remove obsolete\nsetting of html_last_updated_fmt, project, latex_elements from conf.py.\n\nAdd doc8 to lower-constraints to make requirments-check happy.\n\n[1] https://governance.openstack.org/tc/reference/project-testing-interface.html#documentation\n\nChange-Id: If86dd619402495d9d4470b14cb270fcf53db6794\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33efadb678c7c3d0b6a7b66fc9dab8284667b919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33efadb678c7c3d0b6a7b66fc9dab8284667b919"}]},"branch":"refs/heads/master"},"af04cc3392f7361d9537c49510447b63e8802b31":{"kind":"REWORK","_number":6,"created":"2019-08-10 09:42:00.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/27/675727/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/675727/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/675727/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/675727/6"}}},"commit":{"parents":[{"commit":"99ec31f53f5fc4cf62c338a4b2b28b884c2ab623","subject":"Docs: Make links more robust","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99ec31f53f5fc4cf62c338a4b2b28b884c2ab623"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 08:11:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-10 09:41:14.000000000","tz":120},"subject":"Update docs building","message":"Update docs building\n\nCleanup doc/requirements to just use what is needed for doc building.\nMove reno to doc/requirements, add doc8 to test-requirements.\n\nupdate tox.ini:\n* doc8 should be in pep8 as linter and not docs according to PTI [1]\n* Only build main doc in docs environment, CI uses this and we waste\n  just time building api-ref as well, there\u0027s a separate environment for\n  this. This follows PTI [1]\n* Do not include requirements file for docs building, it\u0027s not needed\n  with apidoc.\n* Use common deps for all docs build environments.\n\nCleanup doc/source/conf.py, with the switch to api-doc a lot of settings\nare not needed anymore, also the eventlet bug is fixed.\n\nUpdate openstackdocstheme to 1.20.0 and which allows to remove obsolete\nsetting of html_last_updated_fmt, project, latex_elements from conf.py.\n\nAdd doc8 to lower-constraints to make requirments-check happy.\n\n[1] https://governance.openstack.org/tc/reference/project-testing-interface.html#documentation\n\nChange-Id: If86dd619402495d9d4470b14cb270fcf53db6794\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af04cc3392f7361d9537c49510447b63e8802b31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af04cc3392f7361d9537c49510447b63e8802b31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
