)]}'
{"id":"openstack%2Fcinder~679858","triplet_id":"openstack%2Fcinder~stable%2Frocky~I253123d5451b32f0e3143916e41aaa1af75561c2","project":"openstack/cinder","branch":"stable/rocky","topic":"bug/1812913","hashtags":[],"change_id":"I253123d5451b32f0e3143916e41aaa1af75561c2","subject":"Fix VolumeAttachment is not bound to a Session","status":"MERGED","created":"2019-09-03 20:22:49.000000000","updated":"2019-09-04 15:46:15.000000000","submitted":"2019-09-04 14:54:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679858-1567608885896-86c0bb62","meta_rev_id":"8e26329016984cb6204f1b4a9ed3f0e9eeb27ca5","_number":679858,"virtual_id_number":679858,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-09-03 22:45:59.000000000","_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-03 22:00:37.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-09-03 21:18:38.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":2,"date":"2019-09-04 14:54:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-03 22:45:21.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-03 21:14:26.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-04 03:39:32.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"date":"2019-09-04 02:07:33.000000000","_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-04 04:33:20.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2019-09-03 21:57:04.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2019-09-04 00:20:43.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-03 22:32:35.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-03 23:49:09.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-09-04 02:52:04.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-04 12:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-04 09:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-04 12:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 21:14:26.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-03 21:18:38.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2019-09-03 21:57:04.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2019-09-03 22:00:37.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-09-03 22:32:35.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-03 22:45:21.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-03 22:45:59.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-03 23:49:09.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-09-04 00:20:43.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-04 02:07:33.000000000","updated_by":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-04 02:52:04.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-04 03:39:32.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2019-09-04 04:33:20.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-09-04 09:07:49.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2019-09-04 12:52:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-04 14:54:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb7e2e46ba90b033615cac123305c993d8c44530","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-09-03 20:22:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"451b09b3aae62662d60fb0c3cda3c10278698085","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-09-03 20:22:57.000000000","message":"Topic changed from bug/1812913-stable/rocky to bug/1812913","accounts_in_message":[],"_revision_number":1},{"id":"3a806cb214b422777eacb7b5b92501f6e239148f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-03 21:14:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/58/679858/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/15be7cd : FAILURE in 21m 28s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/58/679858/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a259169 : FAILURE in 23m 15s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"0f86c03fedac4c881e76a748ab4da72fb7e3926b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2019-09-03 21:18:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/679858/1/iscsi/ : FAILURE in 51m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/679858/1/smb/ : FAILURE in 48m 20s","accounts_in_message":[],"_revision_number":1},{"id":"17f3b66e6c59d0513d8ccabad34a589fc75a9e1c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2019-09-03 21:57:04.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/679858/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"077c47ccbfbf279eb2a8eda887c06b1d9c3ce323","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-09-03 22:00:37.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/58/679858/1/check/dsvm-tempest-infinibox-fc/587a88e : FAILURE in 25m 42s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/58/679858/1/check/dsvm-tempest-infinibox-iscsi/2cdfa5d : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9cf582625efeb91361c17eabeca9daa27cc61167","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2019-09-03 22:32:35.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-NEC Cinder CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-ISCSI/58/679858/1/check/NEC-ISCSI/0a3e617 : FAILURE in 38m 25s\n- NEC-FC http://FL9-119-243-99-132.tky.mesh.ad.jp/NEC-FC/58/679858/1/check/NEC-FC/13c40f4 : FAILURE in 35m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1e7885a2f4218b9eb4ed6461dc26406fb2262ae3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 22:44:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d81fc9b7c23d49089a24ed088bd236f0 : SUCCESS in 8m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/918c763ff77e499c9416d4cce728ebf9 : SUCCESS in 10m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b1909b87d17c40fb8850bf2c0feb2978 : SUCCESS in 11m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/65dc0edbd58c4cac8a6dce73defd167e : SUCCESS in 10m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8c7d2e0729d343e3a0dc86fb05443ba2 : SUCCESS in 1h 38m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a82364d7b32e44e89d901c2b133df724 : SUCCESS in 1h 05m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/963a15e78193436c97582f889f0abfe0 : SUCCESS in 1h 46m 45s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ecab69cff5434008b8dc01f6b60a64b8 : SUCCESS in 1h 51m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e4c422398c4846a390f4b32cc9b5c638 : SUCCESS in 8m 45s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/87390b24a00141ef9f3e4ebee56aa217 : SUCCESS in 1h 02m 24s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/49cddaad6ac44631b2669b7dfb5f27c8 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/342ddd1d9d2f4f16aa5a3b18d16fc377 : SUCCESS in 10m 00s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ce7dcd9c0624f4998e74c0b11367b0a : SUCCESS in 5m 17s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/0c3b2f91bd62453686b4de8435c83331 : SUCCESS in 8m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9ecb67e6126741388a003e3a751398d0 : SUCCESS in 40m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b923cc4b7a449f9b9d16228ac9d9b35 : SUCCESS in 30m 21s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5f11a432beb4223bce13d27abb829b8 : SUCCESS in 1h 08m 51s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f51b31eb83794efb9a37f3ffcd532da8 : SUCCESS in 7m 54s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d663d82e49d84fbfa0708ac1a34018b2 : SUCCESS in 1h 03m 51s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e47a37b337ce40d38284e3c588be359b : SUCCESS in 1h 16m 01s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/08d2b6d3f3be430892521600687e03e1 : SUCCESS in 1h 19m 20s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/cb6dbd1440cf47a59c47a1137bb2a150 : SUCCESS in 1h 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91b38df7845f2b4cc2154c9c0d832be99dcc7a81","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2019-09-03 22:45:21.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/58/679858/1/check/synology-dsm-cinder-iscsi/c56684e : FAILURE in 11s","accounts_in_message":[],"_revision_number":1},{"id":"c19951c2c709f760a079ad4360cb4dc4afa3d4e7","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-09-03 22:45:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://13.56.159.105/logs/58/679858/1/upstream-eseries-check/cinder-eseries-iSCSI/4e0c8df/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://13.56.159.105/logs/58/679858/1/upstream-eseries-check/cinder-eseries-FCP/15b37ba/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":1},{"id":"bba232729b282bf65e1951f10f78b6c0179db589","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-09-03 23:49:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/58/679858/1/upstream-solidfire-check/cinder-solidfire-iSCSI/3b7bcd1/ : SUCCESS in 1h 02m 52s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"6dd4e9a55a7e383981f17f0135fcd111c57c5d37","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-04 00:20:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/58/679858/1/upstream-check/cinder-cDOT-iSCSI/7363312/ : SUCCESS in 57m 35s\n- cinder-cDOT-NFS http://13.56.159.105/logs/58/679858/1/upstream-check/cinder-cDOT-NFS/88e708a/ : SUCCESS in 54m 03s\n- cinder-cDOT-FCP http://13.56.159.105/logs/58/679858/1/upstream-check/cinder-cDOT-FCP/e81609b/ : SUCCESS in 1h 35m 00s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"863632e241da377cdaa5c9611eede0451d69cd0c","author":{"_account_id":13144,"name":"Oracle ZFSSA CI","username":"oracle-zfssa-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 02:07:33.000000000","message":"Patch Set 1:\n\n* oracle-zfssa-dsvm-iscsi https://openstackci.oracle.com/zfssa_ci/refs-changes-58-679858-1-iscsi : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6033dec2f887b4cfd74bfb0436c506f990312c46","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2019-09-04 02:52:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/58/679858/1/check/dsvm-tempest-zadara-driver/6bed8e9 : SUCCESS in 1h 48m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88e676d638f968b04a171c905c1fe7894b431120","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2019-09-04 03:39:32.000000000","message":"Patch Set 1:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://us.linbit.com:8080/CI-LINSTOR/58/2670 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d5fe907207195dd1430c1732fff4142b7e528d8e","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-09-04 04:33:20.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/58/679858/1/check/kaminario-dsvm-tempest-full-iscsi/abb8266 : FAILURE in 53m 50s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/58/679858/1/check/kaminario-dsvm-tempest-full-FC/6a852ba : FAILURE in 53m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ac69d79f79e7f58b7602ab06fbfe76cb293f0e47","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-09-04 09:07:49.000000000","message":"Patch Set 1: Code-Review+2\n\nBackport looks good, we will also have to backport Iac785eef9be4b9cdb5c739ee0a87949805282867 to have the fix complete.","accounts_in_message":[],"_revision_number":1},{"id":"d6e87c86f2acb012b940f5fc36b314011a7fe238","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 12:52:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGood to backport. Merged to master when master was stein.","accounts_in_message":[],"_revision_number":1},{"id":"c7b2acaa993c349cc4a3e85875e32b273bfb3eac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:53:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"091e0f88aaead8dc75fe4b5569bb337b99eca26e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:54:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102473b263c3469092c8e07d993e5106 : SUCCESS in 16m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f41288901484678920f961ee88603a9 : SUCCESS in 10m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/65a423583b464e4eac4c1edc170d961f : SUCCESS in 9m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1c943ab91714455ab4c7dc779dc548ac : SUCCESS in 10m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/54efa36ace7d4c6b89464bbc2d237f6b : SUCCESS in 1h 50m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cd8c4e184267478d868b25bee4996ec7 : SUCCESS in 1h 03m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19dfc773170846c39339205e16d99aa2 : SUCCESS in 1h 52m 59s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/7b5afdac2cac45bdad5bb6376fcf14e1 : SUCCESS in 1h 08m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/37075b45545d4108bacbab5c4ab9d305 : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94bd105bb70143ea8e6ce54a528f8cfa : SUCCESS in 11m 15s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1041b7c25b3747eabd19ca5c48759b94 : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"91ac63d9f413b2011fb4259f65a02fb8551eb405","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:54:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e26329016984cb6204f1b4a9ed3f0e9eeb27ca5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-04 15:46:15.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/679858/1/check/tempest-dsvm-full-xenial/fe6bf83/ : FAILURE in 1h 57m 24s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/679858/1/check/devstack-plugin-ceph-dsvm-xenial/8497974/ : FAILURE in 2h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb928c4b3ab7a4086b1a716dc9cea2809fbcf5fe","revisions":{"cb928c4b3ab7a4086b1a716dc9cea2809fbcf5fe":{"kind":"REWORK","_number":1,"created":"2019-09-03 20:22:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/58/679858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/679858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/679858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/679858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/679858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/679858/1"}}},"commit":{"parents":[{"commit":"98196cc765b4e44a1a37a47f2d52ad0d39423bd7","subject":"Merge \"Remove experimental openSUSE 42.3 job\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98196cc765b4e44a1a37a47f2d52ad0d39423bd7"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-01-22 20:24:37.000000000","tz":60},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2019-09-03 20:22:49.000000000","tz":0},"subject":"Fix VolumeAttachment is not bound to a Session","message":"Fix VolumeAttachment is not bound to a Session\n\nIn some cases, when loading a list of volumes using\ncinder.objects.VolumeList.get_all we may end up with a SQL alchemy error\nlike this one:\n\n sqlalchemy.orm.exc.DetachedInstanceError: Parent instance\n \u003cVolumeAttachment at 0x7f048ff22290\u003e is not bound to a Session; lazy\n load operation of attribute \u0027volume\u0027 cannot proceed (Background on this\n error at: http://sqlalche.me/e/bhk3)\n\nThis happens because when loading a Cinder Volume we also load the\nvolume attachments, so the Volume OVO tries to create the\nVolumeAttachmentList OVO, which in turn creates the individual\nVolumeAttachment OVOs, but these have the volume field set in\nexpected_attrs, but in this case the volume data is not loaded.\n\nIn most cases we don\u0027t see this issue because the session is still\navailable and SQLAlchemy will return this information.\n\nInstead of using the dictionary \"get\" method that will do the lazy\nloading we use the hasattr method that will return False if the lazy\nloading exception happens when trying to get the volume.\n\nDepends-On: I9f0fec25444ed865d56d0d250fb6d840ab5b4095\nCloses-Bug: #1812913\nChange-Id: I253123d5451b32f0e3143916e41aaa1af75561c2\n(cherry picked from commit 42ce739a0841e39bd267cc2fd19be53447faac85)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb928c4b3ab7a4086b1a716dc9cea2809fbcf5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb928c4b3ab7a4086b1a716dc9cea2809fbcf5fe"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
