)]}'
{"id":"openstack%2Fcinder~703770","triplet_id":"openstack%2Fcinder~master~Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a","project":"openstack/cinder","branch":"master","topic":"fujitsu-driver-update","attention_set":{},"removed_from_attention_set":{"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2023-07-19 14:47:25.000000000","reason":"Change was abandoned"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2023-07-19 14:47:25.000000000","reason":"Change was abandoned"},"29838":{"account":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"last_update":"2023-07-19 14:47:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a","subject":"Fujitsu Driver: Add QoS support","status":"ABANDONED","created":"2020-01-22 11:11:36.000000000","updated":"2023-07-19 14:47:25.000000000","total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"2bde5b88b417c1522bf27ebe24ca02743c89ea69","_number":703770,"virtual_id_number":703770,"owner":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-05 09:13:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26387,"name":"Qian Mingyue","email":"qianmy.fnst@cn.fujitsu.com","username":"qianmingyue"},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":13689,"name":"Hironori Shiina","email":"Hironori.Shiina@fujitsu.com","username":"shiina"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":-1,"date":"2023-07-12 13:32:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26387,"name":"Qian Mingyue","email":"qianmy.fnst@cn.fujitsu.com","username":"qianmingyue"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-07-05 20:07:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13689,"name":"Hironori Shiina","email":"Hironori.Shiina@fujitsu.com","username":"shiina"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26387,"name":"Qian Mingyue","email":"qianmy.fnst@cn.fujitsu.com","username":"qianmingyue"},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":13689,"name":"Hironori Shiina","email":"Hironori.Shiina@fujitsu.com","username":"shiina"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26387,"name":"Qian Mingyue","email":"qianmy.fnst@cn.fujitsu.com","username":"qianmingyue"},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":13689,"name":"Hironori Shiina","email":"Hironori.Shiina@fujitsu.com","username":"shiina"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":13689,"name":"Hironori Shiina","email":"Hironori.Shiina@fujitsu.com","username":"shiina"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26387,"name":"Qian Mingyue","email":"qianmy.fnst@cn.fujitsu.com","username":"qianmingyue"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 11:11:48.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-02-04 11:49:06.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-04 14:43:00.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-11 08:25:58.000000000","updated_by":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"reviewer":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-11 12:49:16.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-02-12 18:08:16.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2020-03-04 02:06:50.000000000","updated_by":{"_account_id":26387,"name":"Qian Mingyue","email":"qianmy.fnst@cn.fujitsu.com","username":"qianmingyue"},"reviewer":{"_account_id":26387,"name":"Qian Mingyue","email":"qianmy.fnst@cn.fujitsu.com","username":"qianmingyue"},"state":"REVIEWER"},{"updated":"2020-03-17 12:05:31.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2020-03-18 17:18:44.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-03-19 02:11:24.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-03-23 06:19:39.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2020-03-23 08:51:24.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-03-24 10:17:23.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2020-03-24 11:02:17.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-25 12:48:07.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 03:31:23.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"REVIEWER"},{"updated":"2020-04-23 08:08:18.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 09:02:11.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 09:12:22.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 09:18:03.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-04-23 09:45:29.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-04-23 10:10:13.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 11:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 12:04:19.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 12:52:13.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-04-23 13:04:38.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 13:16:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 15:55:17.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 16:08:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-04-23 16:35:03.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-04-23 17:04:43.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-04-23 17:35:07.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 17:51:49.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 23:33:25.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-04-24 03:29:07.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-24 12:20:49.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-24 12:27:02.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-14 04:26:02.000000000","updated_by":{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},"reviewer":{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},"state":"REVIEWER"},{"updated":"2020-05-14 09:22:47.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-15 04:30:29.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2022-07-05 07:07:59.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-07-05 08:31:32.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-05 08:35:54.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-05 08:38:54.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-05 20:07:29.000000000","updated_by":{"_account_id":13689,"name":"Hironori Shiina","email":"Hironori.Shiina@fujitsu.com","username":"shiina"},"reviewer":{"_account_id":13689,"name":"Hironori Shiina","email":"Hironori.Shiina@fujitsu.com","username":"shiina"},"state":"REVIEWER"},{"updated":"2022-07-07 02:36:19.000000000","updated_by":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"reviewer":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"state":"CC"},{"updated":"2022-07-18 15:10:33.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-03-28 01:17:47.000000000","updated_by":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"reviewer":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"state":"REMOVED"},{"updated":"2023-07-12 13:32:53.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-07-12 13:53:35.000000000","updated_by":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"reviewer":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"state":"CC"},{"updated":"2023-07-12 18:00:11.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"a2e3a760df74bad48dbc88f74467d2880af33ca5","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-01-22 11:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"893255ccd55fda99d50f2b70efa6a69c7f0b2b9a","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-01-22 11:11:48.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"3d967e9fc550bb742dfadf4c3b40689dbdfa0c4b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-01-22 11:11:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/70/703770/1/check/DellEMC_XtremIO_FC/5efc727/DellEMC_XtremIO_FC/None : NOT_REGISTERED\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/70/703770/1/check/DellEMC_XtremIO_iSCSI/0a88b40/DellEMC_XtremIO_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"6811ba36ff97aca908232ddddc0ab004d53dbc09","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-01-22 11:13:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"9de895b65dfa2a8848c1656655bdd90b92c60ed9","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-01-22 11:23:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/70/703770/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b9b8ea9 : FAILURE in 10m 27s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/70/703770/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/feb4aa2 : FAILURE in 11m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"ce12b30eb470b94cc4fa52d2ae4cc97fab3e1910","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-01-22 11:30:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/70/703770/1/check/cinder-storpool-tempest/eb921c0/ : FAILURE in 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b667869d200d37d59c49640b7b812996cea56de2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-01-22 11:34:58.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7b174b612d6f930cec3972b5ce2f8fae0682dff8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-01-22 11:35:00.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"71297d011ac988f23f80aa06c75496a25974f3d5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-01-22 11:38:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/1/iscsi/ : FAILURE in 25m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/1/smb/ : FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":1},{"id":"dbdb4a1142eb27991ea7d6e6fec7d6d4311e240f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-01-22 11:41:55.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/70/703770/1/check/dsvm-tempest-infinibox-fc/4493519 : FAILURE in 29m 48s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/70/703770/1/check/dsvm-tempest-infinibox-iscsi/8e019fb : FAILURE in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"30db70e297ba20ea2c5e8b32c3b043038bc8eb8d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-22 11:43:54.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/70/703770/1/check-not-vote/ext-cinder-zuul/6cfd5c7 : FAILURE in 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1de0aa92ed651fc2c9018a50c7a3e345c6326aa3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-22 12:02:50.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/1/check/tempest-dsvm-full-xenial/a0ca014/ : FAILURE in 40m 45s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/1/check/devstack-plugin-ceph-dsvm-xenial/1f6ad4e/ : FAILURE in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c10f7c9b58663f9804ffd4261c1a2a3f57301a4d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-01-22 12:16:14.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fdfdb0ac2d1ef67db75b5b38b42b20a6c0010de5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-01-22 12:25:01.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0ad3037c5daeb8fd435db7709c2b81d96252d6d8","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-01-22 12:41:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-iscsi/c1ce5fc : FAILURE in 1h 21m 05s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-fc/51c00d9 : FAILURE in 1h 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6d9ae874e85a3056e933b493e83748fff3dfed92","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-01-22 13:15:42.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c0a6a59a7529c38acbb49d65bc423398a6e9f7ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 13:52:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fd8fb8d002e457db39dbe51b058205c : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/766689fa9b444602931714a2607d6772 : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/853ef87a2e1249e8870828ec47e00676 : SUCCESS in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b88294800d094a18843744232f727684 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb7181c720f45879ccdc5e4fb084c17 : SUCCESS in 13m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/001c5830db3848a486bb727abe46ad52 : SUCCESS in 1h 00m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9011852aed048e582fb63b32a35f524 : FAILURE in 1h 05m 23s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b34c0d59834c46c6a75ebb70930fbfb8 : FAILURE in 10m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/378f5f047b98454496c814688525a554 : FAILURE in 9m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2a5868b323048db9b7bbd3c98bad22e : SUCCESS in 6m 33s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e9075aee55e94df3a1d205b4bb8d327a : SUCCESS in 7m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7dd1892756a44af6aec834e780139044 : SUCCESS in 43m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f96eb9660cc40eea6f3a703885d5de1 : FAILURE in 4m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9223ccf818a442698cb0ffc7d200873 : SUCCESS in 1h 28m 04s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f97d722a12ae4aab9d11730a10eaa257 : SUCCESS in 1h 32m 56s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4ed9917139c440e696583a5999aea749 : FAILURE in 10m 57s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/7cf6468eba3648dda7dbfe14e8076548 : FAILURE in 12m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94781c0ff7264894bb0df97d392ddf75 : SUCCESS in 2h 27m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9ed01bbcbf44d809ad080891c6049d8 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7dae1f36b5fda28fc331ecd5ffeb1d2ab3a3fde3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-01-22 22:10:26.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-703770-1/ns5_nfs : SUCCESS in 2h 5m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-703770-1/ns5_iscsi : SUCCESS in 1h 54m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ea7586320fd3c0db723127f0f5b724fe96b5a5f2","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-01-27 02:01:01.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"da9f6e5fee10be2cd54b2a35fe5e88513a6a4f23","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-02-04 10:42:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"770b3e70b57a27b86039923fb0ecd5dab3db59ba","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-04 10:44:19.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/1/iscsi/ : FAILURE in 1m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/1/smb/ : FAILURE in 1m 29s","accounts_in_message":[],"_revision_number":1},{"id":"31dce9ba60de45893c8447b5c84befa30e9ac41a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-04 10:57:01.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ed0c8b36d02889c04650cd73162b4d0fe7f14ffe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-02-04 11:15:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : ERROR Failed to update project opendev.org/openstack/nova","accounts_in_message":[],"_revision_number":1},{"id":"ac30441d3a8d9d7edcd0d0452cc44da8794e4163","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-04 11:47:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/70/703770/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b5dd8d3 : FAILURE in 14s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/70/703770/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/53add46 : SUCCESS in 1h 05m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"c4b4be37010bad23b790dc727c0a87821ce585e8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-02-04 11:49:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/70/703770/1/check/dsvm-tempest-zadara-driver/67a4d7d : SUCCESS in 1h 06m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15e274955f97d0ff7bd48e5efddd035b43de72e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-04 12:17:07.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2028/703770/1/check/infortrend-cinder-volume-iscsi/f170d43a341e2d241dd5205a3b3beee6da0cd948 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2012/703770/1/check/infortrend-cinder-volume-fc/f170d43a341e2d241dd5205a3b3beee6da0cd948 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7d156fa8c2351e7085c194c418e7aba5fce0220c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 13:14:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cf7e9d3aa0b4fc4af59e0c839167d4c : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b0e2f1c120d4578b1a8b1cf5bf3beae : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ebbfe7f52d74afe9915c6a82690ecb8 : SUCCESS in 10m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/608b9842f08d46bda8157c529fa51f79 : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c134c862624349aaa9469251a4136f4b : FAILURE in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475357971ddc45e789c61949cc16a5a5 : SUCCESS in 11m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c050491b53fd4b119d6ab0c6589024d6 : SUCCESS in 1h 06m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85df924415f540b3a94f8714e1d80b65 : SUCCESS in 1h 27m 25s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9407ec4ea7fd49e2bd160b9cf003b789 : SUCCESS in 1h 27m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3135763b5c4f48efa2a36346210dbde8 : FAILURE in 7m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d4767b18da84e4ab6b87a468b38b229 : SUCCESS in 6m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2aab3d857d2a4b949bec72f47af242cb : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8a735e4c72f848eb9ef51e4533591cbe : SUCCESS in 39m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7bd1426a1914b7295f6dd5559cef203 : FAILURE in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/22704698136e4206ac1be82c32ff8b53 : SUCCESS in 1h 07m 28s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b005b5a6b8024b1ab5da428e52cfed0f : SUCCESS in 1h 11m 23s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0d9602775e84704b2327b6cd62264c5 : SUCCESS in 1h 16m 53s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/16354f5b53ad456fa5de07a5623bde08 : SUCCESS in 1h 30m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9bc6043f8c9459885961e39042776d6 : SUCCESS in 2h 26m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85be39cce8a146d4aaedeaea5c46c55b : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"83bbd82d03be607cd9352fff8e76742532d9fa1f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-04 13:24:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/70/703770/1/check/dsvm-tempest-cisco-zonemanager-job/5a82396/ : FAILURE in 2h 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7702e2fb45ac7dc17ddc22194db48095a82060a0","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-02-04 13:27:53.000000000","message":"Patch Set 1:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3271 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f2012c5b01f08afdedd643bed8b46b2172584057","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-04 13:40:29.000000000","message":"Patch Set 1:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-70-703770-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"970dde0a18a69072099e969c827ec696c388ad44","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-02-04 14:01:38.000000000","message":"Patch Set 1: Workflow-1\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":1},{"id":"740190208e5f88649fb9a8f1cf8215d2c12c6196","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-04 14:16:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/70/703770/1/upstream-solidfire-check/cinder-solidfire-iSCSI/e599324/ : FAILURE in 1h 00m 53s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"d54117b0f86216d62aa6ba8c0e651e1de2bd0930","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-04 14:19:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/70/703770/1/upstream-check/cinder-cDOT-iSCSI/7a4d982/ : SUCCESS in 49m 36s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/70/703770/1/upstream-check/cinder-cDOT-NFS/56ff6cb/ : SUCCESS in 51m 43s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/70/703770/1/upstream-check/cinder-cDOT-FCP/2997e35/ : FAILURE in 1h 04m 08s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"38615f6aebcf88f29c50307c57fd5f1531992387","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-04 14:32:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/1/check/tempest-dsvm-full-bionic/1d34398/ : SUCCESS in 2h 33m 10s\n- devstack-plugin-ceph-dsvm-bionic finger://zuul-v3-executor.novalocal/586a8e935610467a9d5c31f40ec44d77 : TIMED_OUT in 2h 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d44bde2e99119e1d840e5303c93b5e1613a468f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-04 14:39:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-iscsi/33f4161 : FAILURE in 36m 52s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-fc/1e924c7 : FAILURE in 37m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c567a9a2a2be3af674dadaa49dbcd9d678c5a413","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-04 14:42:33.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/70/703770/1/check/DellEMC_XtremIO_FC/750e193/DellEMC_XtremIO_FC/456 : SUCCESS in 1h 27m 58s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/70/703770/1/check/DellEMC_XtremIO_iSCSI/e6061fd/DellEMC_XtremIO_iSCSI/443 : SUCCESS in 1h 13m 41s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"ab2169d18ae262dab2cbf1b380c1d5fa710a2de4","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-02-04 14:43:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Cinder-AS13000 http://218.29.111.78:9000/70/703770/1/check/Cinder-AS13000/8780978 : SUCCESS in 3h 16m 57s\n- cinder-test-fc http://218.29.111.78:9000/70/703770/1/check/cinder-test-fc/e24f716 : SUCCESS in 1h 04m 22s\n- cinder-test-iscsi http://218.29.111.78:9000/70/703770/1/check/cinder-test-iscsi/ff511d4 : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e86faee7cec2e0cbf84700495e41b85db0788d66","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-02-04 14:51:27.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/70/703770/1/check/DellEMC_VNX_iSCSI/3642a27/DellEMC_VNX_iSCSI/283 : SUCCESS in 1h 36m 52s\n- DellEMC_VNX_FC http://publiclogs.emc.com/70/703770/1/check/DellEMC_VNX_FC/b485f93/DellEMC_VNX_FC/284 : SUCCESS in 56m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"0c8acd2dc3a164a7c46dd44f3550ed833d34b0b4","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-02-04 14:53:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/70/703770/1/check/DellEMC_SC_FC/a22a8f1/DellEMC_SC_FC/288 : SUCCESS in 1h 08m 25s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/70/703770/1/check/DellEMC_SC_iSCSI/5f527d9/DellEMC_SC_iSCSI/289 : SUCCESS in 1h 38m 24s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"5bb17508a5649832defeb91a84ed551a6116f916","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-02-04 14:58:18.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/703770-1/Seagate-iSCSI : SUCCESS in 38:45.37\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/703770-1/Seagate-FC : SUCCESS in 33:56.94","accounts_in_message":[],"_revision_number":1},{"id":"a6f2592c9317552ad4643e31de31045775adc56e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-04 15:14:30.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6d2114452a0b7fc94e69143a1112795dff299e3d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-02-04 15:19:04.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/70/703770/1/check/DellEMC_VxFlexOS/9ccf8dd/DellEMC_VxFlexOS/280 : SUCCESS in 2h 04m 23s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"8316bb4abb89a21d1d9d2a445a3e1fae5268c3e7","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-02-04 15:19:38.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/703770-1/HPMSA-iSCSI : SUCCESS in 47:16.55\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/703770-1/HPMSA-FC : SUCCESS in 46:30.11","accounts_in_message":[],"_revision_number":1},{"id":"9d3a8b3cb0ad6ee49ee724c2792c2fba6bc7b474","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-02-04 15:20:02.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/703770-1/Lenovo-iSCSI : SUCCESS in 48:22.10\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/703770-1/Lenovo-FC : SUCCESS in 46:22.65","accounts_in_message":[],"_revision_number":1},{"id":"5391cd129dc4bed15ea39961d271f786f528814c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-04 15:47:26.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2029/703770/1/check/infortrend-cinder-volume-iscsi/f170d43a341e2d241dd5205a3b3beee6da0cd948 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2013/703770/1/check/infortrend-cinder-volume-fc/f170d43a341e2d241dd5205a3b3beee6da0cd948 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9293c9abc736047bfd9ae2993c1ff7f5517a13b3","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-04 15:57:14.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e28a17e7d717974c95a404dcc9ed609d2bdcf4de","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-04 19:12:06.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e1b3492c8005ce2fa5e6f865c48104734dbe2888","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-02-05 11:20:14.000000000","message":"Patch Set 1:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":1},{"id":"1a5e33ddf2ea246693e9f38ddedcf352eb3db4e5","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-02-05 11:41:55.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/70/703770/1/check/synology-dsm-cinder-iscsi/d9a2388 : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bb5397898f5454701067f7794db85323541f665b","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-02-05 12:35:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/189/index.html : FAILURE in 10m 25s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/329/index.html : FAILURE in 10m 07s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/385/index.html : FAILURE in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f8a7b90e05b82af3bde3bf7c361d750e0a4f28d7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-05 12:49:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-iscsi/5c7a467 : FAILURE in 1h 28m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-fc/fdeff9d : FAILURE in 1h 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"389b4f8679c1a6158a906bc48d720082b555938d","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-02-05 13:11:31.000000000","message":"Patch Set 1:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":1},{"id":"a726fc3aae4fd3f594f23283f0321ec92e9551b1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-05 14:37:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-iscsi/3e561e6 : FAILURE in 1h 23m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-fc/bfba4fe : FAILURE in 1h 24m 56s","accounts_in_message":[],"_revision_number":1},{"id":"853a32ff3dccc4a7a7a2f154b8f208b70ab829be","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-05 14:57:12.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cfb5a8b87f834da02307b2f11996034ca0595de6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-05 17:20:40.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2040/703770/1/check/infortrend-cinder-volume-iscsi/f170d43a341e2d241dd5205a3b3beee6da0cd948 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2025/703770/1/check/infortrend-cinder-volume-fc/f170d43a341e2d241dd5205a3b3beee6da0cd948 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"02dd3285cd59c06818bef30f13f113bd34e88694","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-02-10 03:04:16.000000000","message":"Patch Set 1:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":1},{"id":"10e644444aec2eeb38b663f1fe74cd390d7c3e4a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-10 04:53:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-iscsi/b556523 : FAILURE in 1h 48m 13s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/1/check/fujitsu-eternus-dx-fc/fe1be58 : FAILURE in 1h 43m 42s","accounts_in_message":[],"_revision_number":1},{"id":"da6ac988cd8665fea5901fc4c966241484afcb88","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-10 06:42:37.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8754c4d6e3829f7f40bc4e93f3fe24899271ebc2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-10 14:52:06.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2046/703770/1/check/infortrend-cinder-volume-fc/f170d43a341e2d241dd5205a3b3beee6da0cd948 : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2061/703770/1/check/infortrend-cinder-volume-iscsi/f170d43a341e2d241dd5205a3b3beee6da0cd948 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"76db4c7fe4c5b61e61ba76544b8da6ec97e47a06","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-02-11 08:24:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b68c47e0e0b02e5b2a6085422bedad03f3612ff","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2020-02-11 08:25:58.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"bf9274ebb92ed53b396943109fec7524ac873487","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-02-11 08:35:40.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/70/703770/2/check/nimble-iscsi-driver-dsvm-volume/b98ad42 : FAILURE in 12s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/70/703770/2/check/nimble-fc-driver-dsvm-volume/3698e87 : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":2},{"id":"240724eba8d89f744525919fa76393cfb6658858","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-11 09:14:53.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/70/703770/2/check/DellEMC_XtremIO_FC/df4b312/DellEMC_XtremIO_FC/503 : FAILURE in 46m 37s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/70/703770/2/check/DellEMC_XtremIO_iSCSI/8535137/DellEMC_XtremIO_iSCSI/490 : FAILURE in 41m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"8ca830f6834ecc94043cbe2fc08f7441e01f6a52","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-11 09:25:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/70/703770/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e85a261 : FAILURE in 43s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/70/703770/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/c15e1ab : SUCCESS in 1h 00m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"6ccae4a00a65e10c1bcb850763f9050df159312d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-11 09:30:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/2/iscsi/ : SUCCESS in 1h 05m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/2/smb/ : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a787c0e6aaec5e8bdf31ab396a9a363512b0b2a6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-11 09:32:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/70/703770/2/check/dsvm-tempest-cisco-zonemanager-job/f98b151/ : FAILURE in 59m 30s","accounts_in_message":[],"_revision_number":2},{"id":"bf37fb6371739d8c8b33566fbc0139b549940ec4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-11 09:51:53.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/2/check/fujitsu-eternus-dx-iscsi/335486f : SUCCESS in 1h 21m 40s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/2/check/fujitsu-eternus-dx-fc/7cf5222 : SUCCESS in 1h 19m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d3d2ea5442c331ba527094c9bbc6ad656e766485","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-02-11 10:06:44.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/70/703770/2/check/DellEMC_Unity_FC/bd04d7b/DellEMC_Unity_FC/464 : SUCCESS in 1h 13m 02s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/70/703770/2/check/DellEMC_Unity_iSCSI/6ec62cc/DellEMC_Unity_iSCSI/457 : SUCCESS in 1h 38m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a3c400642a3eede7f4fcf6583caa9e024a30a052","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-02-11 10:26:10.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-703770-2/ns5_nfs : SUCCESS in 2h 1m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-703770-2/ns5_iscsi : SUCCESS in 1h 57m 3s","accounts_in_message":[],"_revision_number":2},{"id":"c5d7309baec8cea8cc89ece5cfaf42e221765d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 10:46:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e852cff4f3441b8b9909d81ddbe7c67 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/400b5f09a65f49149b8963f5ba15afdb : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc0143746b144070a4418423776b30c4 : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49b40f18dace4175b9a5fc54c8d6d56b : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c0fd3e842754d1f937c442ee1caf6b8 : FAILURE in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c722868da2724ef7b335aa7c2a96d936 : SUCCESS in 13m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e7689645cfc94a809d0dfb2a71aa973a : SUCCESS in 57m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06328ccfbb5e4485ae10d91c54f99cac : SUCCESS in 1h 08m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/313658093be3415ebb1ae24ddaa2a9b3 : SUCCESS in 16m 54s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae79ae89ebf54e5e88b64bcfba0a0231 : SUCCESS in 1h 23m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f3a12323fe0349139f17f72cc410eb6f : FAILURE in 7m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb9b39d5808e4f74a4ae3384eda634f5 : SUCCESS in 6m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/19e4c63a850444f983e0d03f8a352f34 : SUCCESS in 7m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b72c64cdd1e943018d1972510bd41c13 : SUCCESS in 36m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f2223cf7600e4cbb8997af2077ddc779 : FAILURE in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d96a4296594643c191e18f91d85b04a8 : FAILURE in 1h 10m 35s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e468223b85dd44578ecde25f62d5413d : SUCCESS in 1h 05m 24s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9a32128edf944ecf8f6fa76b24d6a7ee : SUCCESS in 1h 12m 34s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/5679c126dddf4d7ca85fc7712cc5c7cf : SUCCESS in 1h 16m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5afd024a0ce44e77ba635f51b08c4852 : SUCCESS in 2h 15m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a71de8bc81c4152991144b365bc4b96 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7b322ca5953d63a6ba774550aec89b6a99b6bb50","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-11 10:47:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/2/check/tempest-dsvm-full-bionic/63cb792/ : SUCCESS in 2h 08m 32s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/2/check/devstack-plugin-ceph-dsvm-bionic/95928e7/ : SUCCESS in 2h 14m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0aa90457a3c6e07a7f6660dc4a38dd9ee2ef96e","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-02-11 11:12:34.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/70/703770/2/check/synology-dsm-cinder-iscsi/57ac74a : FAILURE in 25m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6fd1657653099ef326f513f2caaac9a3957ae712","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-02-11 11:24:12.000000000","message":"Patch Set 2:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3293 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4808d52deaac621594a2c389c029ff95ebd9448b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-11 11:49:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/70/703770/2/upstream-solidfire-check/cinder-solidfire-iSCSI/5a19e10/ : SUCCESS in 1h 01m 27s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"3de56995ca63f256522ab2eef433ee1f7a548df7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-11 11:50:17.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-70-703770-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"c974e1a7dcb80a49c1d2cf2949bd6391f02f2772","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-02-11 12:00:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/2/20200211190139/NEC-FC : SUCCESS in 1h 40m 51s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/2/20200211190139/NEC-ISCSI : SUCCESS in 1h 58m 44s","accounts_in_message":[],"_revision_number":2},{"id":"694548b9fa4ba723224eb482c3a19f50cbf66719","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-11 12:22:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/70/703770/2/check/DellEMC_XtremIO_FC/4474a50/DellEMC_XtremIO_FC/505 : SUCCESS in 1h 27m 27s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/70/703770/2/check/DellEMC_XtremIO_iSCSI/cab1fca/DellEMC_XtremIO_iSCSI/492 : FAILURE in 1h 17m 39s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"7e290e47a3525ceb92d89432a68d65c037793f16","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-11 12:24:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/70/703770/2/upstream-check/cinder-cDOT-iSCSI/4b758e8/ : SUCCESS in 1h 35m 55s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/70/703770/2/upstream-check/cinder-cDOT-NFS/cefb69b/ : SUCCESS in 1h 37m 10s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/70/703770/2/upstream-check/cinder-cDOT-FCP/f55af03/ : FAILURE in 1m 09s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"fcfb4b49bad7799091e5f29c559ad7247386e13e","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-02-11 12:33:00.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/703770-2/Seagate-iSCSI : SUCCESS in 38:45.06\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/703770-2/Seagate-FC : SUCCESS in 35:33.42","accounts_in_message":[],"_revision_number":2},{"id":"beb386547b6032e1d6955f70cb3d687ca5ff6af9","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-02-11 12:35:38.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/70/703770/2/check/DellEMC_SC_FC/a4e2141/DellEMC_SC_FC/307 : FAILURE in 1h 06m 28s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/70/703770/2/check/DellEMC_SC_iSCSI/7a47a5e/DellEMC_SC_iSCSI/308 : FAILURE in 1h 48m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"e3570e8733604f09e317e4818c4d44daade3ac22","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-02-11 12:49:16.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/703770-2/HPMSA-iSCSI : SUCCESS in 46:28.43\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/703770-2/HPMSA-FC : SUCCESS in 45:37.72","accounts_in_message":[],"_revision_number":2},{"id":"80013f1b64dbc37b14d78b786eedef75d20e624d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-02-11 12:50:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/70/703770/2/check/DellEMC_VxFlexOS/9b4faf2/DellEMC_VxFlexOS/306 : SUCCESS in 2h 03m 17s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"f0a4955e1da81553b21923558a542ee650db4c55","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-02-11 12:50:59.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/703770-2/Lenovo-iSCSI : SUCCESS in 48:43.56\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/703770-2/Lenovo-FC : SUCCESS in 45:16.43","accounts_in_message":[],"_revision_number":2},{"id":"13325208c73b9ccea85f28d13642f863bd35d1eb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-11 13:17:38.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"12da78ad061be21ffb4279b474ce5617a26808f8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-02-11 13:35:03.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/70/703770/2/check/DellEMC_Unity_FC/bbd3d1c/DellEMC_Unity_FC/466 : SUCCESS in 1h 15m 26s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/70/703770/2/check/DellEMC_Unity_iSCSI/b32ccc3/DellEMC_Unity_iSCSI/459 : SUCCESS in 1h 50m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"67792b4dd74c6e96571ea67d927760db32b87b82","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-02-11 13:42:40.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/70/703770/2/check/DellEMC_VNX_iSCSI/b34e740/DellEMC_VNX_iSCSI/302 : FAILURE in 2h 55m 35s\n- DellEMC_VNX_FC http://publiclogs.emc.com/70/703770/2/check/DellEMC_VNX_FC/484d5f0/DellEMC_VNX_FC/303 : SUCCESS in 56m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"a419b530b46cdcf52def6cd748d93756e873066c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-11 14:45:47.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"13e7c2f6524ef194faa35a4654dfcedce1cb97b4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-11 16:49:54.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"40708190131794e158ede7cd8811798e25f30f44","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-02-12 01:31:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/70/703770/2/check/cinder-storpool-tempest/d2fd712/ : SUCCESS in 2h 20m 14s","accounts_in_message":[],"_revision_number":2},{"id":"aef5ffaa9f779f94336636d41420653d637b274e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-12 18:08:16.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2057/703770/2/check/infortrend-cinder-volume-fc/945efec4a0d49fd8dca3f14e1dc009e1bf79246a : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2072/703770/2/check/infortrend-cinder-volume-iscsi/945efec4a0d49fd8dca3f14e1dc009e1bf79246a : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"68224721c3a87a58f07cd82d278680ff3fe4fea7","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-02-13 06:01:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/227/index.html : FAILURE in 33s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/381/index.html : SUCCESS in 1h 34m 33s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/421/index.html : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0a6bb9a0ba12c46b7b90af15a4f2599f67186297","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-02-15 04:49:22.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/70/703770/2/check/dsvm-tempest-infinibox-fc/4d62141 : EXCEPTION\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/70/703770/2/check/dsvm-tempest-infinibox-iscsi/3718ae5 : EXCEPTION","accounts_in_message":[],"_revision_number":2},{"id":"f739a2e21bf6d51a17635a83d81d8ece2577ef0c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-16 13:47:01.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"6f7b212784679b6dc4bd60a30bd1f17bf75c2c42","author":{"_account_id":26387,"name":"Qian Mingyue","email":"qianmy.fnst@cn.fujitsu.com","username":"qianmingyue"},"date":"2020-03-04 02:06:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b9f53060a7ee290b7fa08b295e77d48584a4f50d","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2020-03-17 12:05:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nminor items","accounts_in_message":[],"_revision_number":2},{"id":"2155b4515a9590ac2eac1a4da32388aba8813d70","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-03-18 08:54:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"20e986bb96d8f9bec0f36973bf8335accc8bfba6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-18 09:28:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-iscsi/88093dc : FAILURE in 32m 23s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-fc/7083328 : FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":3},{"id":"516329ceebe6c693e06fc76417bf05623c14a25d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-18 09:52:04.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/3/iscsi/ : FAILURE in 56m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/3/smb/ : FAILURE in 26m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d3ee06b7991ae089b2f16d359fc013c2c030e05d","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-03-18 09:59:12.000000000","message":"Patch Set 3:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":3},{"id":"49629e0ab2919cc6a86a89b30d8431d53e207a14","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-18 10:27:49.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/70/703770/3/check-not-vote/ext-cinder-zuul/b12708e : FAILURE in 1h 27m 28s","accounts_in_message":[],"_revision_number":3},{"id":"ab7629d53407993b86b43a983760aee8acc5dc12","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-18 10:31:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-iscsi/3900337 : FAILURE in 31m 56s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-fc/f90fdb2 : FAILURE in 32m 00s","accounts_in_message":[],"_revision_number":3},{"id":"234d44970501898c8793eea87b7a7541aa8b019f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-18 10:43:40.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-703770-3/ns5_nfs : SUCCESS in 1h 32m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-703770-3/ns5_iscsi : SUCCESS in 1h 48m 23s","accounts_in_message":[],"_revision_number":3},{"id":"8d56ab231cb7acac124c42c6d43c13eb13cdc302","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 11:12:01.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/70/703770/3/check/DellEMC_XtremIO_FC/89e31c0/DellEMC_XtremIO_FC/783 : SUCCESS in 1h 28m 29s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/70/703770/3/check/DellEMC_XtremIO_iSCSI/41660e0/DellEMC_XtremIO_iSCSI/732 : FAILURE in 1h 56m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"5f889100b7fff2e7c5b4121b4e5471172daf926f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-18 11:18:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/70/703770/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/5f8fc13 : UNSTABLE in 1h 02m 03s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/70/703770/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/9fddc78 : UNSTABLE in 1h 23m 41s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"2fcdd865a24beae7152d9afed37a55af975fc91f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-18 11:41:00.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/02cf5e299c444961ae31db6d80d76e64 : RETRY_LIMIT in 38s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/3/check/devstack-plugin-ceph-dsvm-bionic/dbb6275/ : SUCCESS in 1h 50m 41s","accounts_in_message":[],"_revision_number":3},{"id":"83f485b74d45027756f2622c476ad045a03a588b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 12:03:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/3/20200318182842/NEC-FC : SUCCESS in 2h 31m 49s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/3/20200318182842/NEC-ISCSI : SUCCESS in 2h 34m 43s","accounts_in_message":[],"_revision_number":3},{"id":"8d48bf5db19848c80adad66bb06c07487eede2bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 13:05:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b39c4ea57c24b12acfe4b52fdc62f77 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3408bfe8e7474c19b63290ac13c92715 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b470116b49949e1b03268e4c324189b : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38c1e8a2c2e04b968ca8a7916f6f2552 : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d100fc2b61ee4286ad7fdfa7af935b0d : FAILURE in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2cd53345ef45789518a801dcda2a5e : SUCCESS in 14m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d7d01a6186f74fc6b8be671b57674824 : SUCCESS in 1h 06m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6fd8e1129b4f4083866c1547a8a01510 : SUCCESS in 1h 20m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea201b7dcce841a28a8bdb9251cf8d22 : SUCCESS in 16m 58s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/244afcba0d924b06a6a4180869df382d : SUCCESS in 1h 23m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01cbeab558e54ccaaf0c758bc8d7cfdc : FAILURE in 7m 59s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/14a638aec8db42c49e3950f7c852fa54 : SUCCESS in 5m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d2c9a5f2b0bb48d2be87819830dd845e : SUCCESS in 8m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7594673ae34e4727b583183aca42ea7a : SUCCESS in 44m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fdfc502252074dc2b2d4abfe4e18dc3c : FAILURE in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/093984ea98624af7b16838cefae83470 : SUCCESS in 1h 06m 45s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c0fbac925d648ba91ddf1900e1f7b36 : SUCCESS in 1h 33m 54s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9b25429980dc487b8dbd295a44afe614 : SUCCESS in 1h 37m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/efbc60f92a7f465d9ccc0a426a5eb93e : SUCCESS in 1h 41m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2664a3ec2c024753973ff63234882ffa : SUCCESS in 2h 46m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ede7455f67043bbb97acef27fd5f320 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":3},{"id":"3246ec499e3656cc4e947988b29aecbba53f428a","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-18 13:17:36.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/70/703770/3/check/synology-dsm-cinder-iscsi/dcee75b : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":3},{"id":"cfec76f1589424780bc633cb05fb3e948158498d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-18 13:45:19.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c1037ce9bc7e1fed5fd2d97e80d40c8604c5c6de","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-18 13:57:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/70/703770/3/upstream-check/cinder-cDOT-iSCSI/c0067d4/ : FAILURE in 6m 42s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/70/703770/3/upstream-check/cinder-cDOT-NFS/6fbce69/ : SUCCESS in 46m 43s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/70/703770/3/upstream-check/cinder-cDOT-FCP/4d31a76/ : FAILURE in 16m 51s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"4bf9228c82189c488d1fb946aee9bec4d98a33be","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-18 14:06:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/70/703770/3/upstream-solidfire-check/cinder-solidfire-iSCSI/261b2ec/ : SUCCESS in 58m 59s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"3b88bfbdf127e27273c67b4a905b6b57486a1b4f","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-18 14:10:11.000000000","message":"Patch Set 3:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3484 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"19539e1d79a6441e7134d36cde78acf5be3a17e6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-18 14:10:35.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/70/703770/3/check/DellEMC_Unity_FC/00d0534/DellEMC_Unity_FC/690 : SUCCESS in 1h 31m 04s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/70/703770/3/check/DellEMC_Unity_iSCSI/a862883/DellEMC_Unity_iSCSI/671 : SUCCESS in 1h 59m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b485a5a9d5f8aae8665c1370eb74e6a9e1a34fc8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 14:25:11.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/70/703770/3/check/DellEMC_VNX_iSCSI/0c8b69b/DellEMC_VNX_iSCSI/408 : SUCCESS in 1h 19m 41s\n- DellEMC_VNX_FC http://publiclogs.emc.com/70/703770/3/check/DellEMC_VNX_FC/78573d3/DellEMC_VNX_FC/409 : SUCCESS in 1h 01m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"71e2746fc2adf0157ac3d8df196ff815cc86dd05","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-03-18 14:25:42.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b4740a979bf21b1caf78c3578f5c3681f280134f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-18 14:26:18.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-70-703770-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"56bee70d9fc826134b37676bf00fa860a5a27260","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-18 14:50:53.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"23802d4a8b52ed1111b806844feda0e4e130cdb1","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-03-18 14:52:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/466/index.html : SUCCESS in 1h 12m 35s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/612/index.html : FAILURE in 13s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/652/index.html : SUCCESS in 1h 23m 43s","accounts_in_message":[],"_revision_number":3},{"id":"93dc1012ece79556fe39885794010b4b86a179d2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 16:30:54.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/70/703770/3/check/DellEMC_XtremIO_FC/f7807f9/DellEMC_XtremIO_FC/786 : SUCCESS in 1h 28m 22s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/70/703770/3/check/DellEMC_XtremIO_iSCSI/6235fad/DellEMC_XtremIO_iSCSI/735 : FAILURE in 1h 55m 02s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"972983553715213e08f6c6ccf6c7463569fbed27","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-03-18 16:35:11.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/703770-3/Seagate-iSCSI : SUCCESS in 39:54.33\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/703770-3/Seagate-FC : SUCCESS in 35:39.52","accounts_in_message":[],"_revision_number":3},{"id":"726a585e1afe06afd706427edf49ed8ecfb4874d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-03-18 17:18:44.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/703770-3/Lenovo-iSCSI : SUCCESS in 49:31.23\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/703770-3/Lenovo-FC : SUCCESS in 47:16.74","accounts_in_message":[],"_revision_number":3},{"id":"fda1cfb8674a7516b65803610aca3dc494db8564","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 17:30:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/70/703770/3/check/DellEMC_VxFlexOS/466f235/DellEMC_VxFlexOS/446 : SUCCESS in 2h 10m 03s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"1dfaa55d2ea90aedb0f9f92e2b3057c7d4d3ddf7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-03-18 18:29:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/70/703770/3/check/cinder-storpool-tempest/8f280dd/ : SUCCESS in 3h 04m 46s","accounts_in_message":[],"_revision_number":3},{"id":"55819437dc7fe5861dd749b55ec39405295bac0a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-03-18 19:32:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/70/703770/3/check/DellEMC_SC_FC/7e2fd62/DellEMC_SC_FC/440 : SUCCESS in 1h 09m 56s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/70/703770/3/check/DellEMC_SC_iSCSI/fd3f881/DellEMC_SC_iSCSI/442 : FAILURE in 2h 52m 02s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"7a9dc612bd8454fb350bdc17fbaaf3371a3d3f4e","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-03-19 00:58:34.000000000","message":"Patch Set 3:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":3},{"id":"b57aab49a54d0dd567051de11c77b3fb56d69a62","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-19 01:01:28.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"66b1d7c4870a12662781641114caaa04eed48ba4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-19 02:11:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/70/703770/3/check/dsvm-tempest-infinibox-fc/f88a86f : SUCCESS in 1h 03m 38s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/70/703770/3/check/dsvm-tempest-infinibox-iscsi/eb01c40 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f9415d89d3d49ff24c66661650839dfa67a0181e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-19 03:25:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-iscsi/6721940 : FAILURE in 2h 25m 10s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-fc/c4ddb82 : FAILURE in 2h 02m 17s","accounts_in_message":[],"_revision_number":3},{"id":"1eab8dd4f5651823205fd0d4e5b46b2927b5a3e0","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-03-19 05:24:01.000000000","message":"Patch Set 3:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":3},{"id":"7c57f894d9ea73f871b809d3a846b545ac51c371","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-19 06:04:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-iscsi/3322a97 : FAILURE in 39m 33s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-fc/77fd1d1 : FAILURE in 37m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ad0aade6bd3f0b411a929d3615d2a765b314de7c","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-03-19 15:56:41.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/70/703770/3/check/nimble-iscsi-driver-dsvm-volume/4e668f0 : FAILURE in 2h 13m 15s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/70/703770/3/check/nimble-fc-driver-dsvm-volume/48f0c70 : FAILURE in 1h 28m 06s","accounts_in_message":[],"_revision_number":3},{"id":"5db90aee27970329c9fce8274e07f6df600f41af","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-03-19 22:37:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/70/703770/3/check/kaminario-dsvm-tempest-full-iscsi/5806e3d : SUCCESS in 2h 25m 10s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/70/703770/3/check/kaminario-dsvm-tempest-full-FC/519c5ea : SUCCESS in 2h 30m 44s","accounts_in_message":[],"_revision_number":3},{"id":"b1c119ef1ddf498752f0cbc380d1ded89e3a2d07","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-03-20 00:56:14.000000000","message":"Patch Set 3:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":3},{"id":"b45320f13fb074bf73aa4f55cfe49eafcae178d8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-20 01:07:31.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"ea7d31296b958a21474553a15fe7d3abed914ee2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-20 01:29:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-iscsi/191e1f4 : FAILURE in 32m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-fc/c623973 : FAILURE in 32m 48s","accounts_in_message":[],"_revision_number":3},{"id":"c835d7abf32912fed542a2efc4e6623a949116c5","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-03-20 02:07:54.000000000","message":"Patch Set 3:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":3},{"id":"087ac5d70f03733ace3d8b4c1b1b191a30600289","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-20 02:40:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-iscsi/7bb6c35 : FAILURE in 31m 23s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-fc/c714169 : FAILURE in 31m 39s","accounts_in_message":[],"_revision_number":3},{"id":"3a5aac258abe9366a2a9d56990731c1a6f6b3525","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-21 02:56:33.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"a7b2edadf8d4a550a778ee148ddf76053c8ae011","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-03-23 01:37:03.000000000","message":"Patch Set 3:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":3},{"id":"0333fe1d1ad305f6790489d462292dd39ce7e6af","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-23 03:31:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-iscsi/b5e634d : FAILURE in 1h 53m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/3/check/fujitsu-eternus-dx-fc/df4db81 : FAILURE in 1h 52m 07s","accounts_in_message":[],"_revision_number":3},{"id":"5125430719801228964afd12152b4db623495d86","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-23 05:31:59.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"a12c0f305aeafadd949470199cb13914ad88871f","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-03-23 06:19:39.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/70/703770/3/check/kaminario-dsvm-tempest-full-iscsi/5b8c1af : FAILURE in 3h 51m 57s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/70/703770/3/check/kaminario-dsvm-tempest-full-FC/4c36c3b : FAILURE in 3h 43m 04s","accounts_in_message":[],"_revision_number":3},{"id":"8c16fb858801921ff77d8761108762116984e0e7","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-23 08:51:24.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nFujitsu CI failed","accounts_in_message":[],"_revision_number":3},{"id":"7b706707ad6e07893c9a9dd0000ebfa49053cda5","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-03-24 07:48:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec7c5a22113f5e552605c30ad17ee888b48fd36b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-24 07:48:39.000000000","message":"Patch Set 4:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":4},{"id":"c593044eef7ebf795b0f38554a7fe6d4b3117d06","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-24 08:23:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/4/check/fujitsu-eternus-dx-iscsi/99b0a99 : FAILURE in 34m 36s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/4/check/fujitsu-eternus-dx-fc/154aec7 : FAILURE in 34m 55s","accounts_in_message":[],"_revision_number":4},{"id":"77f7b8359edfa5aad0428da692ac069ee7e6948c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-24 08:50:12.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/4/iscsi/ : FAILURE in 59m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/4/smb/ : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":4},{"id":"d155d42176ec9d2e7f752382c744426dea0232a9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-24 09:58:14.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-703770-4/ns5_nfs : SUCCESS in 2h 9m 50s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-703770-4/ns5_iscsi : SUCCESS in 1h 55m 30s","accounts_in_message":[],"_revision_number":4},{"id":"d7fccb6d5801e43533b5935f15f8ea65449540ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 10:05:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b58dc200ebdc48d7a0cdf4042c499ec6 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e6e3a76dd146fca74f5129c25b2a04 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10a232fb6ac14ba0a26eb0947a10a029 : SUCCESS in 8m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2f1b7516c504a97b6d32524029d010f : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba9f1d5955354d32b01f921253590bbc : FAILURE in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784b3920bafa46899007deb2a9fcbad8 : SUCCESS in 13m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e7a79b10af0e43e580732d5a4dc23063 : SUCCESS in 58m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c75507c40d9042389111acab1b683cd0 : SUCCESS in 1h 16m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c812d89eb7364b149119d400103d2274 : SUCCESS in 15m 52s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7baadfc049a1408bbde225c29723eeba : SUCCESS in 1h 17m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e46a00bd48a441197e6cfcccc2ce94f : FAILURE in 6m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/07a350c66c614ae28604b5b9c8097c99 : SUCCESS in 4m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/33a19338fb624b04a46afd1cba898810 : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c9fc7cfa9bcf4416b52ae88f124bfc81 : SUCCESS in 40m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c26fb89f1dc47fb92a54aa16ec6bd81 : FAILURE in 4m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/29c69925cd0e4bb28af9891b940ced5c : SUCCESS in 1h 04m 17s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4e27c8c2444d41bbbe6c3c6bca7dbc07 : SUCCESS in 1h 09m 54s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b7393d57dd1648319ed6959dbd7954ee : SUCCESS in 1h 16m 40s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/1d406ce677374b989d4a9a1019777953 : SUCCESS in 1h 28m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/77957420186d46909d55d7c6a0cff695 : SUCCESS in 2h 10m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93d39ec93eeb41fcac70758c7bd3cd0e : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":4},{"id":"616e7a52690268b93623e6a608c221423aabbf9d","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-24 10:17:23.000000000","message":"Patch Set 4:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3515 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"313e570348324e898a2c3387504558253524fffa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-24 10:20:02.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/4/check/tempest-dsvm-full-bionic/92570de/ : SUCCESS in 1h 27m 16s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/4/check/devstack-plugin-ceph-dsvm-bionic/cae4de0/ : SUCCESS in 1h 49m 00s","accounts_in_message":[],"_revision_number":4},{"id":"6b4a3b81dd1a894e1df734e9f87341e802ef4afe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-24 10:41:11.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-70-703770-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"4d20b7f72b17c0054dba41f43fe3d91d8e171c89","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-24 11:02:17.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/70/703770/4/check/synology-dsm-cinder-iscsi/864cc31 : UNSTABLE in 56m 15s","accounts_in_message":[],"_revision_number":4},{"id":"688525f7fd795c84d34dee96eea828ae9f533e72","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-24 12:45:27.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/70/703770/4/check-not-vote/ext-cinder-zuul/d97e7fe : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":4},{"id":"6f7227a946467e88f5e7516cb1532176f3e5aa68","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-24 13:00:13.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/70/703770/4/check/DellEMC_VNX_iSCSI/e8b500d/DellEMC_VNX_iSCSI/421 : FAILURE in 2h 54m 22s\n- DellEMC_VNX_FC http://publiclogs.emc.com/70/703770/4/check/DellEMC_VNX_FC/44599ee/DellEMC_VNX_FC/419 : FAILURE in 1h 00m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"5aef0684695de3bca4f83dedd76f052275ff0c6d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-24 13:17:33.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/70/703770/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/cb150a3 : UNSTABLE in 59m 56s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/70/703770/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/5b162d4 : UNSTABLE in 1h 21m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"fc9cc7f26c2070619346bcc5a9c57467fd477a19","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-24 14:54:45.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/70/703770/4/check/DellEMC_VxFlexOS/b65b03b/DellEMC_VxFlexOS/472 : FAILURE in 2h 13m 31s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"c09354a49fbe3fb43885405dbcfc54a122058be4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-24 16:47:57.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/70/703770/4/check/DellEMC_XtremIO_FC/aba876c/DellEMC_XtremIO_FC/835 : SUCCESS in 1h 30m 49s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/70/703770/4/check/DellEMC_XtremIO_iSCSI/f6fef68/DellEMC_XtremIO_iSCSI/777 : SUCCESS in 1h 54m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"b86110a76ef3b463fbef28dcda09c799d1c37c89","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-03-24 18:25:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/4/20200325001927/NEC-ISCSI : SUCCESS in 2h 23m 25s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/4/20200325001927/NEC-FC : SUCCESS in 3h 06m 21s","accounts_in_message":[],"_revision_number":4},{"id":"98c0612f8b76d89cf6b9fd2adee8f5e793844bb3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-24 20:17:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/70/703770/4/check/DellEMC_Unity_FC/821a37b/DellEMC_Unity_FC/742 : SUCCESS in 1h 36m 31s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/70/703770/4/check/DellEMC_Unity_iSCSI/faa91af/DellEMC_Unity_iSCSI/719 : SUCCESS in 2h 28m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"31f45b6ab93efb8a64ca1625aaf36ec2c02bc24b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-25 00:58:35.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a208e33e8e3665c918dcd91871f50692d7e4e8a3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-25 02:29:48.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"18cae7f06d5599de8f2eb4a5263be7a8df1809b3","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-03-25 11:14:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/520/index.html : SUCCESS in 1h 08m 48s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/666/index.html : SUCCESS in 1h 35m 39s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/707/index.html : SUCCESS in 1h 24m 29s","accounts_in_message":[],"_revision_number":4},{"id":"25e318103816dfc172ae773e78eeeac69dee479c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-03-25 12:48:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/70/703770/4/check/cinder-storpool-tempest/7ce9c1d/ : POST_FAILURE in 3h 07m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4bec7f02be81b3bc6aa35e371c4e7889ee497602","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-12 14:52:14.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"c603922556d5c390ccd323108abee28751db7943","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-04-23 01:54:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4f95b92ed1bfcb8618e5b1973a1df960990ea8d7","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-04-23 02:01:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f04c8313e73070bab6a70311495156ef13dae1e7","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-04-23 02:05:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"380c514dd1883d021e6822c670197400a52fc34c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-23 02:47:14.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"380699a06b27c31286e937b81120efe076d50367","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-23 02:47:23.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/70/703770/7/check-not-vote/ext-cinder-zuul/0dbd354 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":7},{"id":"875cc310253ba6a9f48eefbe9e1d6190b6b2f446","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-23 02:53:52.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"c4091038ffc7f020b54d7aeb73c9f7157f7ad749","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-23 02:57:50.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"52ba34efa28ddaf9129e201e472503e0ecda002e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-23 03:11:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/7/iscsi/ : SUCCESS in 56m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/7/smb/ : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":7},{"id":"33d9b12dcc6d88885c44e815cc6428926916c4de","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-23 03:30:00.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/7/check/fujitsu-eternus-dx-iscsi/e8eb22d : FAILURE in 1h 14m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/7/check/fujitsu-eternus-dx-fc/0f85c0f : FAILURE in 1h 21m 54s","accounts_in_message":[],"_revision_number":7},{"id":"4fc9e96e3298fab924da870898f28e3f76c4e224","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-23 03:31:23.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/97/index.html : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"96aa6ab5196b6b070f2c0e9cc7f8b2972688aa3b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-23 03:31:31.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/95/index.html : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"3d1a3ce972a8f3762c1cb1fa232c7fc2706ae461","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-23 03:58:00.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/7/check/tempest-dsvm-full-bionic/dbb48bb/ : SUCCESS in 1h 36m 58s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/7/check/devstack-plugin-ceph-dsvm-bionic/990e158/ : SUCCESS in 1h 47m 53s","accounts_in_message":[],"_revision_number":7},{"id":"520a506c3bd2b684eccf822b58463d5ffba52b24","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-23 03:59:17.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/70/703770/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/56a9d02 : FAILURE in 1h 53m 24s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/70/703770/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/2580910 : SUCCESS in 1h 20m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"f4c26d699d47ac450c5ddab33331ee0587aa91e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-23 04:08:48.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/96/index.html : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"d3fb39f581796685b275c7ad9762e0c0e668e2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 04:20:58.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32370b91694e4ee19dadf10c3c6e7df8 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64f5b8501994e7287a443981fee53c7 : FAILURE in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45daa3484de34c3d9034891640ef3e53 : SUCCESS in 9m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b7bb5e4b776472eb243b0864cd6f3d2 : SUCCESS in 16m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2385d11f12f144caa8988ea126b385f7 : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6e6de211df4d35beebd77cc3c2eb48 : SUCCESS in 21m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5502fe4319e84610aa0a27f7270fa391 : SUCCESS in 1h 37m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c78f14735f64e0386bdd196db6ddeac : FAILURE in 1h 44m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bce5cbafd68a4150b51a91e7326c0de4 : SUCCESS in 17m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b431a6c2e7314f909a65d2594304a2af : FAILURE in 11m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b69d1546e480480e9e83475d81c714b1 : SUCCESS in 6m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/adea4325a30c432c8b3969a1263b4ea9 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/681b0278054e4b2bb8f4a9c4f46e7abd : SUCCESS in 36m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81afc6fe2fb74ab089efc3304271a46a : FAILURE in 7m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f53e9f180f7b423b9ba505f7cc40886f : SUCCESS in 56m 42s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/646add30589a4a4d801433a01bf1c0d0 : TIMED_OUT in 2h 05m 29s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/72da4150917f4c7491a25d2482c91005 : SUCCESS in 1h 12m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/19f742d553cc44afa6f1de9f2f558c8e : SUCCESS in 1h 47m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e9846288aea34a51b922cd11017ea12b : SUCCESS in 1h 16m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36afaa2d4b404a5b8013d573e8d99165 : SUCCESS in 2h 10m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f16563cd7e154e60b0832819b55d7fcb : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":7},{"id":"89018e43dc42a6427800facf9e909dd84fa21656","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-04-23 05:22:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"094b565b4136c63fcd94c9c40b67271ec02bdf69","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-23 06:02:10.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/70/703770/8/check-not-vote/ext-cinder-zuul/6a12835 : SUCCESS in 38m 38s","accounts_in_message":[],"_revision_number":8},{"id":"fc4149c0b5b68a4bf8905bea0e3e875b45783cfc","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-23 06:15:02.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"7c4c78b06352908c7df3e93915ff52a722b0b44e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-23 06:18:07.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-703770-8/ns5_nfs : FAILURE in 42m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-703770-8/ns5_iscsi : FAILURE in 54m 59s","accounts_in_message":[],"_revision_number":8},{"id":"4a05f70c07eb358a32fecb0a367a63df4dbdae90","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-23 06:21:31.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/8/iscsi/ : SUCCESS in 58m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/8/smb/ : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":8},{"id":"c240ebd3d6e936f16a1033e0fb417b2662f01858","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-23 07:16:19.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/70/703770/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/0eb6876 : FAILURE in 1h 53m 04s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/70/703770/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/642e32b : SUCCESS in 1h 09m 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"39dc4f155471a0c5f01c4b169be5b8d480449fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 07:54:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/599a44b8780d4d3989a75925516358c3 : FAILURE in 37m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48c35737c66c42e4ab5de538b507916d : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c9095c5dbc041b9ab60ceb3246988ca : SUCCESS in 8m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7ee8fe700764df8ba19af8eec014dbf : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc70928e409a4845a84e75e25283f2f2 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9a46b8a4c245f78774bbfcc1e7f836 : SUCCESS in 11m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ba86040806d743f0888b151def27aa00 : SUCCESS in 1h 13m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/897c308bb37645eaa602deb9d26a9425 : SUCCESS in 1h 21m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34114472d669421c8d9374a412cb2dae : SUCCESS in 15m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6515d6c60a7f4f44946c2e2f4d50b3cc : FAILURE in 7m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8012cf5bce8a4f4989f65412c3dc2888 : SUCCESS in 5m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d01f7c5b2b7e49b5b0f8b08736963497 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/825735be9dcb46edb6939de341297231 : SUCCESS in 39m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e815c4aa60aa4a819c799937da5ae337 : FAILURE in 4m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f775ffbd920a4e2c8260fe997fa30bfe : SUCCESS in 1h 14m 12s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7febd81e111a4cc28d7d30ab0c6bd273 : SUCCESS in 1h 52m 14s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d806244a3c50491cb3921ac958ec3cc2 : SUCCESS in 1h 21m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ff653b5c9d6c409780b4059974675f1b : SUCCESS in 1h 18m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c16a4fc8da1149dfa6812722c2356d0f : SUCCESS in 1h 25m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/61b1a1ce01bd443a8352b9300157f87e : SUCCESS in 2h 26m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e16898d593e84769bab1114d78e3097d : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":8},{"id":"02433aba62bb13a1bf25c721bf0b7caf44f74788","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-23 08:08:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/70/703770/8/check/nimble-iscsi-driver-dsvm-volume/86e4836 : SUCCESS in 1h 21m 39s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/70/703770/8/check/nimble-fc-driver-dsvm-volume/82b72bd : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":8},{"id":"fe4bbb587c18ccf492449b5ff53692805d4eb994","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-23 08:14:06.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/8/check/tempest-dsvm-full-bionic/1e3adf8/ : SUCCESS in 1h 31m 34s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/8/check/devstack-plugin-ceph-dsvm-bionic/0f4eef0/ : SUCCESS in 1h 43m 35s","accounts_in_message":[],"_revision_number":8},{"id":"28644c414fc42b1a769e2bde84ee629b0e749c55","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-04-23 08:20:10.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"94f6780e867797890de457c8f10e33f23c9ff286","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-23 09:02:11.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/70/703770/8/check-not-vote/ext-cinder-zuul/844620d : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":8},{"id":"71211546dcf44d72253373e1d0c153d49292a817","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-23 09:12:22.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/703770/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"d6c2a9ff0709203274f40d7fa932cc7baced6fa5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-23 09:18:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/8/iscsi/ : SUCCESS in 57m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/8/smb/ : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":8},{"id":"6ca780568ff66b86e545c6f45a93671e7f013b15","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-23 09:45:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/70/703770/8/check/dsvm-tempest-cisco-zonemanager-job/e1f910e/ : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":8},{"id":"f909269f17f44431175db7987546862901191326","author":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"date":"2020-04-23 09:50:27.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"651cb40b94ca271f3a47fe61594de6676e1808c0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-23 10:10:13.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/70/703770/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/79b621f : FAILURE in 1h 49m 44s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/70/703770/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b53d6b1 : SUCCESS in 1h 10m 18s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"4724871b7bdce73557ad4ff8f41a9a7e5a20b0bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:51:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e69570e8e84749ea893b8470a2d30dcf : SUCCESS in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daaf2d14803742f383346d26789d1bb8 : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fca7cfec0096412a8997b1c50e79ea65 : SUCCESS in 10m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbe787d1e85b4fedbe320dca9c6ee02c : SUCCESS in 12m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a36bb0c693d044e0a21cc7862d323c21 : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047342c875f14bbfbad41573ce1327db : SUCCESS in 15m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f593b4e9947e4cc193340030a725f009 : SUCCESS in 1h 15m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bd80eaf26fd434e82e161f2508a5b17 : SUCCESS in 1h 15m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba6ce2f1c7be41c9a56e01d1d5d9be28 : SUCCESS in 18m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd6b5db4b5064859bdd2e8cbdf3927f9 : FAILURE in 10m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a94b9567470649fd808dfb854faeb14f : SUCCESS in 9m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ad6eda631d394f399ff15685ae4ce3f1 : SUCCESS in 9m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/10499643ea31499692e62418db168227 : SUCCESS in 42m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa04b12cd4be41569117c61862f5dd73 : FAILURE in 7m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e9ffaede54b48a2a72fd659019a9fe7 : SUCCESS in 1h 13m 53s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/47eb2aa4734f4f41865b50d5613d8bd3 : SUCCESS in 1h 45m 39s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b6cf98ae905b412d987413ba7e53b18b : SUCCESS in 1h 24m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1ffe5429078244e8ad24a0e850416db6 : SUCCESS in 1h 15m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aa9d9b3c65c947e7b0d397fd8acca1b9 : SUCCESS in 1h 16m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/05c4d1d8fb9e45459a1237fafb85bfa1 : SUCCESS in 2h 21m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf585a854e164198a76c19a06e104836 : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":8},{"id":"cc919475e12951fc99e9bd9c1926c33e325ef66d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-23 11:58:54.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-703770-8/ns5_nfs : FAILURE in 2h 2m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-703770-8/ns5_iscsi : FAILURE in 2h 22m 55s","accounts_in_message":[],"_revision_number":8},{"id":"a8b1f295aa3ac89f037b3681f4086dde0ec0b5d0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-23 12:04:19.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/8/check/tempest-dsvm-full-bionic/ccb09ac/ : SUCCESS in 1h 50m 02s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/8/check/devstack-plugin-ceph-dsvm-bionic/ead6f21/ : SUCCESS in 1h 53m 34s","accounts_in_message":[],"_revision_number":8},{"id":"e6579faa6ed3244f58a388d502099b8f12df7810","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-04-23 12:52:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/70/703770/8/upstream-solidfire-check/cinder-solidfire-iSCSI/c5825f3/ : SUCCESS in 59m 55s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"fc18fa672a0473a74426c2267c462ff1cbe0a222","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-23 13:04:38.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/70/703770/8/upstream-check/cinder-cDOT-iSCSI/cc1a6d7/ : SUCCESS in 50m 41s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/70/703770/8/upstream-check/cinder-cDOT-iSCSI-IPv6/d8d0904/ : SUCCESS in 1h 11m 49s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/70/703770/8/upstream-check/cinder-cDOT-NFS/0aab6bb/ : SUCCESS in 49m 53s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/70/703770/8/upstream-check/cinder-cDOT-NFS-IPv6/4b3def5/ : SUCCESS in 1h 12m 37s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/70/703770/8/upstream-check/cinder-cDOT-FCP/77f099a/ : SUCCESS in 1h 03m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"6a4d0c081c02bbd096eb3903a1d9236f64928d77","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-23 13:16:07.000000000","message":"Patch Set 8:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-70-703770-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"2faa6b40689a1bbcb21093ac7839a22152cc941e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-23 15:55:17.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/70/703770/8/check/DellEMC_XtremIO_FC/eb7b126/DellEMC_XtremIO_FC/1148 : SUCCESS in 1h 31m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/70/703770/8/check/DellEMC_XtremIO_iSCSI/62e9a07/DellEMC_XtremIO_iSCSI/1039 : FAILURE in 2h 18m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"8c99bba12a7ef9e0079f1d864a15ca30339d58d6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-23 16:08:18.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/70/703770/8/check/DellEMC_Unity_FC/82139d1/DellEMC_Unity_FC/1042 : SUCCESS in 1h 36m 15s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/70/703770/8/check/DellEMC_Unity_iSCSI/70dd7fe/DellEMC_Unity_iSCSI/993 : SUCCESS in 2h 01m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"4753adbdc6101d6a31c877c537e84b3058f98d6d","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-04-23 16:35:03.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/703770-8/Seagate-iSCSI : SUCCESS in 46:50.97\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/703770-8/Seagate-FC : SUCCESS in 42:16.56","accounts_in_message":[],"_revision_number":8},{"id":"56c5c3b7c57acb419a1f969dac7b1aa120b5b668","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-04-23 17:04:43.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/70/703770/8/check/DellEMC_SC_FC/dfe137b/DellEMC_SC_FC/670 : SUCCESS in 1h 08m 06s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/70/703770/8/check/DellEMC_SC_iSCSI/cc84f2d/DellEMC_SC_iSCSI/669 : SUCCESS in 2h 06m 29s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"6eb5af6468498e23b2523ca67f574ed176b8f0c8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-23 17:35:07.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/70/703770/8/check/DellEMC_VxFlexOS/be7ebf4/DellEMC_VxFlexOS/679 : FAILURE in 2h 19m 44s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"058801e6b04bb127961dcf2675a067da2ae64de8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-23 17:51:49.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/70/703770/8/check/DellEMC_VNX_iSCSI/e981d60/DellEMC_VNX_iSCSI/583 : SUCCESS in 2h 27m 44s\n- DellEMC_VNX_FC http://publiclogs.emc.com/70/703770/8/check/DellEMC_VNX_FC/8872d06/DellEMC_VNX_FC/584 : SUCCESS in 53m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"5894aa20cf48afbd565b41aeca7cf637c19959f2","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-23 23:33:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/809/index.html : SUCCESS in 1h 14m 11s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/958/index.html : SUCCESS in 1h 37m 35s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/997/index.html : SUCCESS in 1h 22m 18s","accounts_in_message":[],"_revision_number":8},{"id":"347afce551f4d83d7267a79628cb87355dc584ba","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-24 03:29:07.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-703770-8/ns5_nfs : SUCCESS in 2h 33m 20s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-703770-8/ns5_iscsi : SUCCESS in 3h 59m 8s","accounts_in_message":[],"_revision_number":8},{"id":"282abcc15a65083901e13748b72c91f3e24c271c","author":{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},"date":"2020-04-24 06:44:09.000000000","message":"Patch Set 8:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":8},{"id":"12090be747916d800702f3283a51b968138fa0f6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-24 09:43:20.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/8/check/fujitsu-eternus-dx-iscsi/9243e56 : FAILURE in 40m 02s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/8/check/fujitsu-eternus-dx-fc/413ccce : FAILURE in 39m 29s","accounts_in_message":[],"_revision_number":8},{"id":"2cab0aa5484d619c8eaf5dcfd7dcf35f737e44ef","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-24 12:20:49.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/70/703770/8/check/DellEMC_PowerMAX_FC/e699e3a/DellEMC_PowerMAX_FC/708 : ABORTED in 1h 11m 37s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/70/703770/8/check/DellEMC_PowerMAX_iSCSI/c8f35ce/DellEMC_PowerMAX_iSCSI/605 : ABORTED in 3m 52s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"8a712bc47b09b634fd273feba2b800c10114c6e8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-04-24 12:27:02.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/8/20200424183709/NEC-FC : SUCCESS in 2h 33m 13s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/8/20200424183709/NEC-ISCSI : SUCCESS in 2h 49m 49s","accounts_in_message":[],"_revision_number":8},{"id":"dc76bcfd8df13737ef129eaba2cf8f960086137b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-24 14:02:13.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"c47f36559bab1c7650c63d121ce61979c9a6045f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-24 15:42:17.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"387a2619dbd4e53a0ca2d43a26774d209e79469f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-24 15:53:34.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"4e5059a89335ec417d96becd9fcf8adba9410496","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-24 17:30:18.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"e71c0ea9c571eba846acfc08a0aad9ddf42d4539","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-24 17:40:22.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"2967964b0499468dd05dbce7e617426c14489980","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-24 19:25:42.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"49685b61d6262ae9fe9caa27653081b8ec9925ff","author":{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},"date":"2020-05-11 02:28:33.000000000","message":"Patch Set 8:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":8},{"id":"4418bc6ed88e8254dfbc698f5493572002d05889","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-05-11 07:34:16.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/8/check/fujitsu-eternus-dx-iscsi/af88c6d : FAILURE in 35m 48s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/8/check/fujitsu-eternus-dx-fc/0e28901 : ABORTED in 5h 00m 01s","accounts_in_message":[],"_revision_number":8},{"id":"2d6930256757f73e27207e43b3782ae7e56f4ea1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-11 09:32:07.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"fcc701848c141dd8400daa39016c2f56e72b4c56","author":{"_account_id":31889,"name":"a_sakaguchi","email":"sakaguchi.akiko@jp.fujitsu.com"},"date":"2020-05-14 04:26:02.000000000","message":"Patch Set 8:\n\nrun-Fujitsu ETERNUS CI","accounts_in_message":[],"_revision_number":8},{"id":"57a1588e824fd348fb4e3cd5c8b667780f8567e7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-05-14 09:22:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/703770/8/check/fujitsu-eternus-dx-iscsi/a7d4c32 : FAILURE in 4h 55m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/703770/8/check/fujitsu-eternus-dx-fc/1670974 : FAILURE in 4h 26m 22s","accounts_in_message":[],"_revision_number":8},{"id":"00267cdad316f4c18f984ed5b459884d7ddddc09","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-15 04:30:29.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/70/703770/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"eb000056f7d87383acee9c0d334a207fc491a625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-05 07:04:42.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"9d3c032b2de65ff922a1a3e9a0656c9cde86c68c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-05 07:07:59.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657004690937 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"e3bd0278842babf987ea831c539206359969e7b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 07:16:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbd431e213644f8aa36d6a492f7a7a67\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da619f74cd8547e19ec631f114699ef0 : FAILURE in 9m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/156eb2b3a9e64e439804d0ade5e274ac : FAILURE in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d06c1b3b34ad64a3cce8ace14aa502a68b3aa248","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-05 07:19:06.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/703770/9/iscsi/ : FAILURE in 13m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/703770/9/smb/ : FAILURE in 13m 53s","accounts_in_message":[],"_revision_number":9},{"id":"1edf90939b936f4ec11f776991e01a59fd2cc1e2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-05 07:39:26.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/9/check/tempest-dsvm-full-focal-py3/b73c252/ : FAILURE in 29m 19s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/70/703770/9/check/devstack-plugin-ceph-dsvm-focal-py3/92c7569/ : FAILURE in 31m 25s","accounts_in_message":[],"_revision_number":9},{"id":"c7e8983da96cc3ff4f48c8048f4b33e69c6a5910","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-05 08:04:55.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/70/703770/9/check/dsvm-tempest-cisco-zonemanager-job/815fa53/ : FAILURE in 51m 09s","accounts_in_message":[],"_revision_number":9},{"id":"4aca795a5780c83cdbd04406c6fd0baddb1c53ab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-05 08:31:32.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/703770 : FAILURE in 1h 25m 09s","accounts_in_message":[],"_revision_number":9},{"id":"6d1b983a40bb815860736f361a504105e02fd326","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-05 08:35:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/703770/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8db830c/ : SUCCESS in 1h 30m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/703770/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7a0ba06/ : SUCCESS in 1h 28m 45s\n- pure-devstack-cinder-tempest-nvme-fc-aio http://openstack-logs.purestorage.com/70/703770/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-fc-aio/57c7a69/ : FAILURE in 28m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/70/703770/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/65798d5/ : FAILURE in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5a8efb6e832b790d9c37434d799d1c55030bb3e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-05 08:38:54.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/70/703770/9/check/iscsi-driver-nimble/d0da459/d0da459.html : FAILURE in 1h 33m 01s","accounts_in_message":[],"_revision_number":9},{"id":"7e9bcb71f7a9d14f3e0c6e29a220af46d89b24d1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-05 09:04:43.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/703770/9/check/DellEMC_Unity_FC/db3bf0a/DellEMC_Unity_FC/1089 : SUCCESS in 1h 15m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/703770/9/check/DellEMC_Unity_iSCSI/5027f3e/DellEMC_Unity_iSCSI/1022 : SUCCESS in 1h 59m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"0e1f5118094284fc64e1652d43e63ef5d3c44b1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 09:13:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aeb1cad2200345a5b307bcd774db1d77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6687a3eeeff149cab7d17a14e372e74f : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df3fd5032a204002b0ac7c6eb0b944a0 : FAILURE in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bf79d436c38475995f8614e8f98e08d : FAILURE in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/706730140e5a41d3bb0ece392ae0259c : FAILURE in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a05b3652a8341ad893072995875ecc1 : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/c9e755ddea8042c38cdd610cd067884e : SUCCESS in 42m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fff9861712a641caa5d1fe968d7c8253 : SUCCESS in 1h 07m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b402d59ec2c486ca040917fe5faebcf : SUCCESS in 36m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/871621e3d4ad460aa44d7eee68582d23 : SUCCESS in 7m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/840ef881499b4b789bc515e5a5d57fb8 : FAILURE in 5m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/04cf7c6654594d15a38f1b6381c181a6 : SUCCESS in 5m 29s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36c9cf89e1a74a8e835f5356edebc690 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c02c5c2292384159b03a198da3e6b655 : SUCCESS in 5m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11a7ce1bca3f4038b7796bd4f0e75121 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7782afea8ddb49cea7e09b91990024f9 : SUCCESS in 30m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10a0ffb8165f4a7db3cd4e36457340bd : FAILURE in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/434fb51690034c58bc13f1294f9a8293 : SUCCESS in 1h 05m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a88006c7307c44ed8bc497f9322388cc : FAILURE in 1h 36m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c6ff434540554a49a3ce7a1c406b3773 : FAILURE in 1h 25m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/40f9dd14e0cf435cb98550d4fd2d33b4 : SUCCESS in 1h 02m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a08911ede0e746fba748c7406e0d2cab : SUCCESS in 1h 12m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c1ecf76883ab4f34bf625de6e3f161d4 : SUCCESS in 41m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f4e77e59cd984356bf64986e3b5b62ec : SUCCESS in 1h 26m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb922145b0b2426b8f4726a2568573a0 : FAILURE in 1h 59m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f3019fa72c94bfab42269a92af1e938 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":9},{"id":"f36078a42d1945625c7bb8a413e3428937354021","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-05 10:40:14.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-703770-9/ns5_nfs : SUCCESS in 3h 33m 43s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-703770-9/ns5_iscsi : SUCCESS in 3h 35m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"dabe24b05e6717c7225cf2e1ec32daddaa23363c","author":{"_account_id":13689,"name":"Hironori Shiina","email":"Hironori.Shiina@fujitsu.com","username":"shiina"},"date":"2022-07-05 20:07:29.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0931087b5cd5041f7769812667323e9d38992c7b","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-07 02:36:19.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0b16c2331ab41655e256cb39f490bfadf83931ce","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-18 15:10:33.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/9/20220718215636/NEC-V-FC : SUCCESS in 1h 56m 43s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/9/20220718215636/NEC-V-ISCSI : SUCCESS in 2h 13m 55s","accounts_in_message":[],"_revision_number":9},{"id":"4cb8a35245b742ce1c6dbd2ef7a07b846c4d524e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-22 18:09:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/9/20220722233745/NEC-FC : SUCCESS in 3h 30m 54s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/703770/9/20220722233745/NEC-ISCSI : SUCCESS in 3h 31m 57s","accounts_in_message":[],"_revision_number":9},{"id":"d58ad80fb468fd733beaf81c760b564b2b411d0b","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-03-28 01:17:47.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_33609\u003e.","accounts_in_message":[{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"}],"_revision_number":9},{"id":"3104dabba95830eceef59e2f2225eff963776b88","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-07-12 13:32:53.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"01278ed662ffdf403ec44ae1f51dd7aeea46c9c3","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-07-12 13:53:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9e27722c1bcd048dc570ee49f5512469d90590df","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-12 18:00:11.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/703770/9/fc : SUCCESS in 0h 55m 11s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/703770/9/iscsi : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":9},{"id":"2bde5b88b417c1522bf27ebe24ca02743c89ea69","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-19 14:47:25.000000000","message":"Abandoned\n\nthis patch is superseded by https://review.opendev.org/c/openstack/cinder/+/847730","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"bc30ede9423373cc3336dc30657bcca6c5e14df7","revisions":{"f170d43a341e2d241dd5205a3b3beee6da0cd948":{"kind":"REWORK","_number":1,"created":"2020-01-22 11:11:36.000000000","uploader":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"ref":"refs/changes/70/703770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/1"}}},"commit":{"parents":[{"commit":"a4e6da9a9371299fbd36c647b89dc12630e7c130","subject":"Merge \"Fujitsu Driver: Change the calculation of TPP\u0027s capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4e6da9a9371299fbd36c647b89dc12630e7c130"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 11:09:21.000000000","tz":480},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f170d43a341e2d241dd5205a3b3beee6da0cd948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f170d43a341e2d241dd5205a3b3beee6da0cd948"}]},"branch":"refs/heads/master"},"945efec4a0d49fd8dca3f14e1dc009e1bf79246a":{"kind":"REWORK","_number":2,"created":"2020-02-11 08:24:32.000000000","uploader":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"ref":"refs/changes/70/703770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/2"}}},"commit":{"parents":[{"commit":"465e62ca6d7c282844b5e9e2105ec2eabed08497","subject":"Merge \"3PAR: Revert flag for Active/Active High Availability Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/465e62ca6d7c282844b5e9e2105ec2eabed08497"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-02-11 08:08:42.000000000","tz":480},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/945efec4a0d49fd8dca3f14e1dc009e1bf79246a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/945efec4a0d49fd8dca3f14e1dc009e1bf79246a"}]},"branch":"refs/heads/master"},"8c384285a686ce1d97273150adad8f996816100f":{"kind":"REWORK","_number":3,"created":"2020-03-18 08:54:55.000000000","uploader":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"ref":"refs/changes/70/703770/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/3"}}},"commit":{"parents":[{"commit":"3f3b7189dba60a2b1fd336241ac8839cd09098ea","subject":"Merge \"Tests: Remove Python 2 compat file_spec code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f3b7189dba60a2b1fd336241ac8839cd09098ea"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-03-18 08:51:54.000000000","tz":480},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c384285a686ce1d97273150adad8f996816100f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c384285a686ce1d97273150adad8f996816100f"}]},"branch":"refs/heads/master"},"1bece2034fd0fd425db71cff772318320c911c9e":{"kind":"REWORK","_number":4,"created":"2020-03-24 07:48:06.000000000","uploader":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"ref":"refs/changes/70/703770/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/4"}}},"commit":{"parents":[{"commit":"3f3b7189dba60a2b1fd336241ac8839cd09098ea","subject":"Merge \"Tests: Remove Python 2 compat file_spec code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f3b7189dba60a2b1fd336241ac8839cd09098ea"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-03-24 07:01:10.000000000","tz":480},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bece2034fd0fd425db71cff772318320c911c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bece2034fd0fd425db71cff772318320c911c9e"}]},"branch":"refs/heads/master"},"9d5c3d7bdcf252341b59fc448bbaefcdac28def1":{"kind":"REWORK","_number":5,"created":"2020-04-23 01:54:58.000000000","uploader":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"ref":"refs/changes/70/703770/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/5"}}},"commit":{"parents":[{"commit":"5ee7c1d6cb9775601ec3d3350aff7d1db47612e6","subject":"Merge \"PowerMax Driver - RDF status validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ee7c1d6cb9775601ec3d3350aff7d1db47612e6"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-04-23 01:51:53.000000000","tz":480},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d5c3d7bdcf252341b59fc448bbaefcdac28def1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d5c3d7bdcf252341b59fc448bbaefcdac28def1"}]},"branch":"refs/heads/master"},"20350a7ab243bc768e8fa2b3d1aead85236373a4":{"kind":"REWORK","_number":6,"created":"2020-04-23 02:01:38.000000000","uploader":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"ref":"refs/changes/70/703770/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/6"}}},"commit":{"parents":[{"commit":"5ee7c1d6cb9775601ec3d3350aff7d1db47612e6","subject":"Merge \"PowerMax Driver - RDF status validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ee7c1d6cb9775601ec3d3350aff7d1db47612e6"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-04-23 01:57:35.000000000","tz":480},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20350a7ab243bc768e8fa2b3d1aead85236373a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20350a7ab243bc768e8fa2b3d1aead85236373a4"}]},"branch":"refs/heads/master"},"f78b6d0031c05f2f0b508cd40041c5c35d2c1fed":{"kind":"REWORK","_number":7,"created":"2020-04-23 02:05:30.000000000","uploader":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"ref":"refs/changes/70/703770/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/7"}}},"commit":{"parents":[{"commit":"5ee7c1d6cb9775601ec3d3350aff7d1db47612e6","subject":"Merge \"PowerMax Driver - RDF status validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ee7c1d6cb9775601ec3d3350aff7d1db47612e6"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-04-23 02:04:54.000000000","tz":480},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f78b6d0031c05f2f0b508cd40041c5c35d2c1fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f78b6d0031c05f2f0b508cd40041c5c35d2c1fed"}]},"branch":"refs/heads/master"},"ec192875a6724ae798933f265af297a6cff35c44":{"kind":"REWORK","_number":8,"created":"2020-04-23 05:22:50.000000000","uploader":{"_account_id":29838,"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","username":"whfnst"},"ref":"refs/changes/70/703770/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/8"}}},"commit":{"parents":[{"commit":"5ee7c1d6cb9775601ec3d3350aff7d1db47612e6","subject":"Merge \"PowerMax Driver - RDF status validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ee7c1d6cb9775601ec3d3350aff7d1db47612e6"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-04-23 05:17:17.000000000","tz":480},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec192875a6724ae798933f265af297a6cff35c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec192875a6724ae798933f265af297a6cff35c44"}]},"branch":"refs/heads/master"},"bc30ede9423373cc3336dc30657bcca6c5e14df7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-07-05 07:04:42.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/703770/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/703770/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/703770/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/703770/9"}}},"commit":{"parents":[{"commit":"ce4d531707d181c39952075b30e06919870f379a","subject":"Merge \"PowerMax Driver - Manage volume into correct storage group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce4d531707d181c39952075b30e06919870f379a"}]}],"author":{"name":"Wu Heng","email":"wuh.fnst@cn.fujitsu.com","date":"2020-01-22 06:31:48.000000000","tz":480},"committer":{"name":"XuQi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-05 07:04:42.000000000","tz":0},"subject":"Fujitsu Driver: Add QoS support","message":"Fujitsu Driver: Add QoS support\n\nAdded support for QoS in the Fujitsu driver.\n\nThe supported qos specs are:\n* maxBWS\n* read_bytes_sec write_bytes_sec total_bytes_sec\n  read_iops_sec write_iops_sec total_iops_sec\n\nChange-Id: Ifaf0cd6fa35e5771ae5d97aff366f199f7638e9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc30ede9423373cc3336dc30657bcca6c5e14df7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc30ede9423373cc3336dc30657bcca6c5e14df7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
