)]}'
{"id":"openstack%2Fcinder~707390","triplet_id":"openstack%2Fcinder~master~I3f2ec1ddf1867261253190953f7a65fff22c7404","project":"openstack/cinder","branch":"master","topic":"bp/vxflexos-replication-support","hashtags":[],"change_id":"I3f2ec1ddf1867261253190953f7a65fff22c7404","subject":"Add OpenStack volume replication v2.1 support in VxFlex OS driver","status":"MERGED","created":"2020-02-12 13:44:28.000000000","updated":"2020-04-14 15:51:10.000000000","submitted":"2020-04-08 23:01:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707390-1586386884714-7fd7f7c7","meta_rev_id":"a2ebca3167c02f07b345ed36db0305d088ff2589","_number":707390,"virtual_id_number":707390,"owner":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2020-04-03 17:09:36.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"date":"2020-04-14 15:51:10.000000000","post_submit":true,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"date":"2020-04-03 22:48:58.000000000","_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"date":"2020-04-03 17:38:48.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"value":0,"date":"2020-04-08 19:33:45.000000000","_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"date":"2020-04-03 16:27:59.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"date":"2020-04-04 00:48:07.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 19:44:36.000000000","_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31416,"name":"wuqiongdan","email":"wuqiongdan@szsandstone.com","username":"wuqiongdan"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"value":0,"date":"2020-04-05 17:02:07.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"date":"2020-04-04 10:00:31.000000000","_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2020-04-04 20:49:45.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-03 16:27:26.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 23:01:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-04 16:35:14.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-04-03 16:46:13.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2020-04-08 06:09:03.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},{"value":0,"date":"2020-04-04 01:00:16.000000000","_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-04-03 16:01:33.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2020-04-03 22:30:49.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-05 06:06:18.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2020-04-06 21:24:12.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-05 23:25:34.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"date":"2020-04-04 05:16:38.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-08 19:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-07 14:44:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-08 19:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31416,"name":"wuqiongdan","email":"wuqiongdan@szsandstone.com","username":"wuqiongdan"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-07 14:40:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-08 19:42:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31416,"name":"wuqiongdan","email":"wuqiongdan@szsandstone.com","username":"wuqiongdan"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31416,"name":"wuqiongdan","email":"wuqiongdan@szsandstone.com","username":"wuqiongdan"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":31416,"name":"wuqiongdan","email":"wuqiongdan@szsandstone.com","username":"wuqiongdan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 13:46:39.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2020-02-12 13:46:58.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2020-02-13 11:49:29.000000000","updated_by":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"reviewer":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-02-13 13:47:40.000000000","updated_by":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"reviewer":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-13 14:19:30.000000000","updated_by":{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},"reviewer":{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},"state":"REVIEWER"},{"updated":"2020-02-13 20:44:04.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-02-19 13:24:27.000000000","updated_by":{"_account_id":31416,"name":"wuqiongdan","email":"wuqiongdan@szsandstone.com","username":"wuqiongdan"},"reviewer":{"_account_id":31416,"name":"wuqiongdan","email":"wuqiongdan@szsandstone.com","username":"wuqiongdan"},"state":"REVIEWER"},{"updated":"2020-02-28 02:55:38.000000000","updated_by":{"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},"reviewer":{"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},"state":"REVIEWER"},{"updated":"2020-02-28 10:30:00.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-01 20:42:23.000000000","updated_by":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"REVIEWER"},{"updated":"2020-03-04 08:57:05.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-03-04 08:59:31.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"},{"updated":"2020-03-04 09:49:45.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-03-04 10:06:21.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-05 13:14:43.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2020-03-19 12:34:22.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2020-03-21 20:22:38.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2020-03-23 12:50:06.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2020-03-31 16:30:54.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 09:28:18.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-04-01 11:32:26.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-04-01 14:57:26.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-04-01 22:07:25.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-02 10:33:04.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-02 15:44:29.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 16:01:33.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 16:27:26.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 16:27:59.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-04-03 16:46:13.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 17:09:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 17:38:48.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 22:30:49.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 22:48:58.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2020-04-04 00:48:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-04 01:00:16.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-04-04 05:16:38.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-04-04 10:00:31.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-04-04 16:35:14.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-04-04 20:49:45.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-05 06:06:18.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-04-05 17:02:07.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-05 23:25:34.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-06 21:24:12.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-07 14:40:57.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-04-07 14:44:40.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-04-08 06:09:03.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-04-08 19:18:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-08 19:42:39.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-04-08 19:44:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 23:01:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-14 15:51:10.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d13cdffef9037d22493c03f2c090138098526393","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-12 13:44:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11492f5103eec0ae094cea2c84c3c83d4411e9e6","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2020-02-12 13:52:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e1c2393b68cfc851a8ff4e03f97aed7221f1c761","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-02-12 13:55:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/1/check/nimble-iscsi-driver-dsvm-volume/91ccf76 : FAILURE in 12s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/1/check/nimble-fc-driver-dsvm-volume/2bda70f : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7813afa9702ec6e9f5debba29845d0af035ae548","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-02-12 13:56:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1885 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1675 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1713 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a131fd31a1071139df52fcab16ee1c800d41c929","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-12 14:00:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/707390/1/check/dsvm-tempest-cisco-zonemanager-job/1b156dd/ : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5afc677c41e6b82e318f390fae733d1e345704f7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-02-12 14:44:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/90/707390/1/check/cinder-storpool-tempest/bf81b22/ : FAILURE in 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b7f3be71ebc2cc571b49ad5eed634ff3799df16f","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2020-02-12 14:48:17.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7653c8c32d213b6d31d38a987790c39f169bfaee","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-12 14:52:48.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/1/iscsi/ : FAILURE in 1h 07m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/1/smb/ : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":1},{"id":"05374e2cd5f6c9b48f4efda04ceaa11cce51a417","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-12 15:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c96819e : SUCCESS in 53m 45s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/30efacc : SUCCESS in 1h 00m 40s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"091e2dab48efa748c78a73703db09aa215325ad7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-02-12 15:35:24.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-1/ns5_nfs : SUCCESS in 1h 49m 36s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-1/ns5_iscsi : SUCCESS in 1h 50m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a729f3a00f429882eabfef43537160df18673ae1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-12 15:44:11.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/1/check/fujitsu-eternus-dx-iscsi/48c6414 : SUCCESS in 1h 21m 50s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/1/check/fujitsu-eternus-dx-fc/970e1fe : SUCCESS in 1h 25m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9baf8779d1853018155ea5490d168caf94cc9f37","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-12 15:53:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/1/check/DellEMC_XtremIO_FC/bbb59eb/DellEMC_XtremIO_FC/517 : FAILURE in 48m 38s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/1/check/DellEMC_XtremIO_iSCSI/32b69e0/DellEMC_XtremIO_iSCSI/504 : FAILURE in 1h 34m 22s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"82ede2505a409e85ce8eec00e7214f5008e2e120","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-12 16:00:52.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/1/check/tempest-dsvm-full-bionic/815206b/ : SUCCESS in 1h 53m 24s\n- devstack-plugin-ceph-dsvm-bionic finger://zuul-v3-executor.novalocal/3d310e73ea9648a986d54000661e0ae7 : POST_FAILURE in 2h 02m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a89a3a3a781ba2f86457b7c17f447a9d876fc434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 16:51:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/638b5346492d4b67bdef91d21666f674 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab28aa47782e43e8bafee037c46e3deb : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/096316042d39401da883fd6e27bf1a60 : SUCCESS in 10m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a58690803434cbcb9cf70b93cf56a53 : SUCCESS in 11m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36e97bc7ecf14213811fe1dc1682a1b9 : FAILURE in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7512237c6dd4b898ffee08470dbd423 : SUCCESS in 16m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e59e36983f074af1b4fd5b717a674ab8 : SUCCESS in 1h 06m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef45d3f3915a49f8a40d960f98b14bfc : SUCCESS in 1h 15m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ed01d4cfc3548ebaafd9ca30e4c606f : SUCCESS in 21m 31s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7da969d800cd46df89b6365ba5af8131 : FAILURE in 1h 17m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3992064ec826430194b64d20eb35a6b5 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2310b73cfaf84f7b860388aff54c66a5 : SUCCESS in 7m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c5d58fb6f9c0420cbf053ad25c223690 : SUCCESS in 9m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a904987d708455aae6f248a988816e4 : SUCCESS in 47m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8822406f8ec4ecd89f235ffe96cb62e : SUCCESS in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/73b26093a9a340e281f53e37f059725a : TIMED_OUT in 2h 10m 06s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5b4aa5fb52a4805ae7476580f11acb6 : SUCCESS in 1h 10m 34s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a7aff64acd1f45e2b00699234cd22ac8 : FAILURE in 1h 30m 45s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/90e746c94e654c5c88e379ba60871efc : SUCCESS in 1h 18m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b197c12eec484fdbbe69eb28ae8b9252 : SUCCESS in 2h 38m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd63abb3e6cf47cfb5cadbf10b5c4605 : SUCCESS in 1h 12m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3faf8def7f1c045bb8cbd5770b0b70f8caada7ca","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-02-12 17:18:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nNITs\nUnit Tests ?","accounts_in_message":[],"_revision_number":1},{"id":"a71f84a4474fadb77b8ff5be805f850692537245","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-02-12 17:39:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/1/20200213001822/NEC-FC : SUCCESS in 2h 08m 25s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/1/20200213001822/NEC-ISCSI : SUCCESS in 2h 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3c481c3a8b546c196ce8b826097f8aad9ee2cde7","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-12 17:49:45.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e NITs\n \u003e Unit Tests ?","accounts_in_message":[],"_revision_number":1},{"id":"d6c3353afce7285fff744a13170de0b8c7ab168a","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-12 17:52:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6fe0e18c1782bd9fa46e6783d4853f6995f50423","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-02-12 17:59:12.000000000","message":"Patch Set 1:\n\n@Ivan shall we add new unit tests for the replication feature?","accounts_in_message":[],"_revision_number":1},{"id":"40448faffa0ebb4b78f28ff3fb4c84d7a2a8fc65","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-12 18:08:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/707390/1/check/dsvm-tempest-cisco-zonemanager-job/845d3aa/ : FAILURE in 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7045e7c8e8f2e580c81b5f6c53e489c48a0f7703","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-12 18:09:44.000000000","message":"Patch Set 1:\n\n\u003e @Ivan shall we add new unit tests for the replication feature?\n\nThere is no so much logic that should be tested, but some tests will not be redundant.\nWill try to add new tests tomorrow.","accounts_in_message":[],"_revision_number":1},{"id":"7296628e87c7f7faefa3ca45ed61c900c0494159","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-12 18:50:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/1/iscsi/ : SUCCESS in 57m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/1/smb/ : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3cd5ed3a92be311590d05f5c3c9fd251d5da940b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-12 18:56:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4288c41 : SUCCESS in 51m 37s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/add3dd2 : SUCCESS in 1h 04m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"b001da713e72094397bc9004eefe43b8df98127a","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-12 18:57:23.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"df52804316497a4a066dfafe5fa2eec20f249b74","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-02-12 19:33:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/90/707390/1/check/cinder-storpool-tempest/fefabb0/ : SUCCESS in 1h 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1dc8bf39105faf7e164111b557be2b2b203ad0ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 20:28:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/128e95e97ea34ddebc56e99429676565 : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c28313beb344d0bd80d38d34e011e1 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e02cfc29b1f4e6b860d6b9a69eb2f22 : SUCCESS in 10m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/244e8bcb4cf34aefb1e10d5cd3c0b0d0 : SUCCESS in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d30d58b28ea43cca7faeddf26fef74b : FAILURE in 12m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ce8e4df5404f1c8b4deada7299be2c : SUCCESS in 18m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2865a8e82b0b4ba1aae610fbdf35690f : SUCCESS in 1h 05m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/093e6208f000444184d6ffad5af7f8e3 : SUCCESS in 1h 25m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/414f21221b4445eaa5ac12d8078d56ec : SUCCESS in 22m 36s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dd9cc774689e45bcbdee1e8c3d3b7cf5 : FAILURE in 1h 45m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a2b9ed1632f40b7abca389c7feb3a9c : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7bf74d3e9b74f78a039241ee98d2fd5 : SUCCESS in 6m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/deacca4b9684489f9e4e044b5ee1b24f : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3b1fd921462144a5b152ce1a36c42a42 : SUCCESS in 46m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1bfc7a631b1b4196ba78d9b67c630e29 : SUCCESS in 7m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63711153e6bd40c9bc6821ea581e97b0 : SUCCESS in 1h 27m 11s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8815983d740643e196aaab975881ce06 : FAILURE in 53m 14s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3c00cc61b7f54d1aa280fa53f39cba7a : SUCCESS in 1h 28m 57s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/6c46663a17fc419a91b9856fd8252279 : SUCCESS in 1h 29m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08441114bcd742e88558895bc967396f : SUCCESS in 2h 29m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94ee43b498ab41618348b142f4d87876 : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6d170a9abe67c39a87897847f6fde053c41412b5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-12 20:53:11.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/1/check/tempest-dsvm-full-bionic/8216686/ : SUCCESS in 2h 04m 54s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/1/check/devstack-plugin-ceph-dsvm-bionic/94f55b6/ : SUCCESS in 2h 06m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ac803659d8a06a58083cddae84ebc1556635328","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-12 23:37:21.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ac41ec0941951ca52655d255f0e282522690ed1b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-13 01:20:17.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0062f8f0ed7437f81bb7231a3edf4401f9e4df69","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-13 03:09:22.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e15b95b4b42ecbc1e9709b0ac8f51be9c170ec57","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-13 07:23:52.000000000","message":"Patch Set 1:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":1},{"id":"7dcb430ce2a133ddcffbf874e66bef6af21e4e19","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-02-13 07:28:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/1/check/DellEMC_VxFlexOS/7028684/DellEMC_VxFlexOS/312 : FAILURE in 4m 37s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"d5bd2ef56a1e6f46a8ea8376204d8d2268e41272","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-13 11:21:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"987e2caa98c54d7f812249d60846f5590d82adca","author":{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},"date":"2020-02-13 11:25:39.000000000","message":"Patch Set 2:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":2},{"id":"8e3243042053d8bdbb0cbdb56b37c00ec4fb13ca","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-02-13 11:32:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/2/check/nimble-iscsi-driver-dsvm-volume/a27d227 : FAILURE in 12s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/2/check/nimble-fc-driver-dsvm-volume/875f3bc : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":2},{"id":"051af83ccca6596e954c3a359acb95d3988ff024","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-02-13 11:33:05.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1894 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1684 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1722 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"c5954005623b410e3bbb1c76b0609c0dc9fac961","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2020-02-13 11:34:56.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"f6a57db7d5101eb20423ec1ea0aa4ffd503522b6","author":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"date":"2020-02-13 11:36:11.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027datacore: recheck\u0027. For information about DataCore CI see https://wiki.openstack.org/wiki/ThirdPartySystems/DataCore_CI\n\n- dsvm-tempest-datacore-iscsi-driver http://173.221.96.158/90/707390/2/check/dsvm-tempest-datacore-iscsi-driver/e5b07f3 : FAILURE in 6s\n- dsvm-tempest-datacore-fc-driver http://173.221.96.158/90/707390/2/check/dsvm-tempest-datacore-fc-driver/f785411 : FAILURE in 4s","accounts_in_message":[],"_revision_number":2},{"id":"94173440587c23a9805620af7ab1eeeb1bc35eba","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-13 11:43:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3531d55e08ffe95e4ad06788c3637b38368ee245","author":{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},"date":"2020-02-13 11:44:29.000000000","message":"Patch Set 3:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":3},{"id":"c6e3d965e1f77b11e5e874704b991423ea137194","author":{"_account_id":24496,"name":"DataCore CI","email":"michael.latchmansingh@datacore.com","username":"datacore-cinder-ci"},"date":"2020-02-13 11:49:29.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027datacore: recheck\u0027. For information about DataCore CI see https://wiki.openstack.org/wiki/ThirdPartySystems/DataCore_CI\n\n- dsvm-tempest-datacore-iscsi-driver http://173.221.96.158/90/707390/3/check/dsvm-tempest-datacore-iscsi-driver/0903d93 : FAILURE in 1m 07s\n- dsvm-tempest-datacore-fc-driver http://173.221.96.158/90/707390/3/check/dsvm-tempest-datacore-fc-driver/1251987 : FAILURE in 3s","accounts_in_message":[],"_revision_number":3},{"id":"f0fa98de8f022c79ddad30e8e05293192bb17bdf","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-02-13 11:55:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/3/check/nimble-iscsi-driver-dsvm-volume/fc9c9f4 : FAILURE in 12s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/3/check/nimble-fc-driver-dsvm-volume/f492e5b : FAILURE in 10m 52s","accounts_in_message":[],"_revision_number":3},{"id":"bac0e3211aa8e02346b6d173515db9285e5b5eb7","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-02-13 11:55:49.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1895 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1685 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1723 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"8782a7c07dba9254e60282276b7ca81db48b0175","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-13 12:44:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/3/iscsi/ : SUCCESS in 59m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/3/smb/ : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":3},{"id":"97984767929638b7e244666a71015e900b22a30e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-13 12:52:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/707390/3/check/dsvm-tempest-cisco-zonemanager-job/56b5224/ : FAILURE in 56m 14s","accounts_in_message":[],"_revision_number":3},{"id":"da12805bf184d71c01725feb629dd730139278f6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-13 13:15:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/3/check/DellEMC_XtremIO_FC/cbf6010/DellEMC_XtremIO_FC/528 : FAILURE in 45m 49s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/3/check/DellEMC_XtremIO_iSCSI/0430374/DellEMC_XtremIO_iSCSI/513 : FAILURE in 1h 07m 31s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"454351426a21df556405badb27dd7eaf5673544e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-02-13 13:17:12.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-2/ns5_nfs : SUCCESS in 1h 55m 43s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-2/ns5_iscsi : SUCCESS in 1h 40m 1s","accounts_in_message":[],"_revision_number":2},{"id":"423949697e3e43f479540b493e55038e788570db","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-13 13:27:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/8617fd3 : FAILURE in 1h 42m 56s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/5e92998 : FAILURE in 1m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"3f908a0c303283309742f61971708e8b18ff8465","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-02-13 13:38:55.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-3/ns5_nfs : SUCCESS in 1h 54m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-3/ns5_iscsi : SUCCESS in 1h 54m 47s","accounts_in_message":[],"_revision_number":3},{"id":"f2d39d08861ac0c0b24431405967d9c6ddb563ef","author":{"_account_id":30590,"name":"SandStone Storage CI","email":"yangziguan@szsandstone.com","username":"yangziguan","tags":["SERVICE_USER"]},"date":"2020-02-13 13:47:40.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- SandStone-iSCSI http://47.75.113.2/90/707390/3/check/SandStone-iSCSI/a76e952/ : RETRY_LIMIT in 9m 36s","accounts_in_message":[],"_revision_number":3},{"id":"5a3d87637c4db3d7e049e90433a8713e7cfb5787","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-13 13:56:36.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/3/check/fujitsu-eternus-dx-iscsi/0e09b0c : SUCCESS in 1h 18m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/3/check/fujitsu-eternus-dx-fc/e126e75 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8eb1651d13fafad3938022f067b40eeca628b6d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:13:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2975f16c9004a9db03e0f363a420741 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a892be8e79434334964477b29afeaecc : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c2f9a379ec845bca62f423e8effeb36 : SUCCESS in 16m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a0d458549e148b39efbcb0d915244aa : SUCCESS in 18m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/803f0b8bae6244ed9bb5e9136900b9f3 : FAILURE in 18m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440c5b6368714b01a078fa84e77aa525 : SUCCESS in 24m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/32f8a8af2dfa4f2982d910968693a8c4 : FAILURE in 1h 20m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e408bdde880d4267bd8b8b4d3485434a : SUCCESS in 1h 57m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/735dc835363b4152bd9357f2c3c34d9b : SUCCESS in 18m 55s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/42b736ed03a84b6bb4ecce2f182a252d : SUCCESS in 2h 25m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/480300aa21424423906cb065e5ce23ed : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ece8029acd20458eaba6635fc497ae11 : SUCCESS in 12m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b178cf04d27d4544a75562ca3a53bc3b : SUCCESS in 14m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2d6e36f2f8094819849b8179f406d81e : SUCCESS in 51m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1fe6a47bfd284fb0ab69e26a1c308722 : FAILURE in 11m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91a5683b651a448180233aef7618f74b : SUCCESS in 1h 13m 54s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b3e23c760331493e8aced3d4ffba9db5 : SUCCESS in 1h 37m 20s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5629a46d9ba340f0b7ab5da33965c9ee : FAILURE in 1h 35m 58s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/fe340326fd0c4a5798f1d75fb0f70880 : SUCCESS in 2h 23m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a89cd3a15ca14cd1bc1467e1a81cccf3 : SUCCESS in 2h 19m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d3b6c2aabf44c40abe3c231e309c0e3 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":3},{"id":"fa88b1309961aba7ae90347b11e4a025f5896a1e","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-13 14:18:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"45603e20190310f4949aaf57d2fbbe046aafa64f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-02-13 14:19:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/707390/3/check/dsvm-tempest-zadara-driver/5cfffde : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"947bd09819e892e8b8cb9010a50cd5044724ae6b","author":{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},"date":"2020-02-13 14:19:30.000000000","message":"Patch Set 3:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":3},{"id":"08ae4b31b992b0e7da796d4a70b53213bcad5b39","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-13 14:26:11.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/3/check/tempest-dsvm-full-bionic/6526b63/ : SUCCESS in 1h 27m 49s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/3/check/devstack-plugin-ceph-dsvm-bionic/3a618d1/ : SUCCESS in 1h 38m 09s","accounts_in_message":[],"_revision_number":3},{"id":"03acd8bd6fd0547e7945d881963305ff40a7e4f1","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-02-13 14:44:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"694929a53daab41cf64869b26b28e3bf71d9e050","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-13 15:34:08.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/3/iscsi/ : FAILURE in 27m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/3/smb/ : FAILURE in 1h 14m 19s","accounts_in_message":[],"_revision_number":3},{"id":"3fb4ba46417facd9b71eecc9a3aff5665ad597ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:43:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7af25509f924150919bb2df854d77b1 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1871544acb7a4d68b21f286f0864e0b0 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42b7e7de944b45058db8831cb2a5b97d : SUCCESS in 9m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56b02d1e99d643bf99492bab2780890c : SUCCESS in 12m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4338e37f974640039ae8fca642bcb689 : FAILURE in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5904ce0be1401aa1c107e0858907c0 : SUCCESS in 15m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a47d826058ce44c79a9c67f34d19bd9b : SUCCESS in 55m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd3cec9b489f4a5d8dcb1f1583536796 : SUCCESS in 1h 22m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7f983a864ee4ca898f52be80a83d23a : SUCCESS in 18m 11s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/480a06d01db6495bbc6c7fd96a7d9c2d : SUCCESS in 1h 42m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/49e21f3811b94d4db9ba3600642d2d68 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2985b63cead4dcdb43e9734c6b42967 : SUCCESS in 6m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a176b60cf14943d0b0d899e75e52e41d : SUCCESS in 7m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/412bfda0b06c4fcd8961af788dd2f799 : SUCCESS in 41m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa9eabaf08b84a9d8f57223472ce8168 : FAILURE in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e3fc8b4cd6714ecfa75933bf4cbf524b : SUCCESS in 1h 10m 09s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/66f44bc8deec4ffbbc0c3e7352189c8a : SUCCESS in 1h 06m 42s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/01d1c1a103d2418ea6d0c6fc0b2ac453 : SUCCESS in 1h 18m 23s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/7348f26f868f46cd84210aea9fb6a436 : SUCCESS in 1h 16m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d52e58e2634e4dae9f65ce2f69dcce2e : SUCCESS in 2h 14m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c66c536bf0340198ed55f218c7ec563 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":3},{"id":"b4e5e5b6b6841406aebff79babaccceff8288c5a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-13 16:52:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/968b02b : FAILURE in 20s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/a7c9bf9 : SUCCESS in 1h 01m 40s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"1fd15eb1e1c1510e90b74f46289da0c55fe8dfb0","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-13 17:34:25.000000000","message":"Patch Set 3:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":3},{"id":"21a678a475b7f5ac1d432eda6a3a53fbe672e6de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-13 17:49:39.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-90-707390-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"f9296e21618c5052669479917f06cd7386f657ae","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-13 18:13:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/90/707390/3/upstream-check/cinder-cDOT-iSCSI/6bc0cfb/ : SUCCESS in 58m 41s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/90/707390/3/upstream-check/cinder-cDOT-NFS/fb4b89c/ : SUCCESS in 56m 01s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/90/707390/3/upstream-check/cinder-cDOT-FCP/e62bd5c/ : SUCCESS in 1h 27m 50s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"8cfb77ff4a989d07a8884ce4784ae61fd11a389f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-13 18:25:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/707390/3/check/dsvm-tempest-cisco-zonemanager-job/d092237/ : FAILURE in 55m 07s","accounts_in_message":[],"_revision_number":3},{"id":"12bbcb408272fd8ac1b089b463c43c0882a14069","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-02-13 18:48:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/3/check/DellEMC_VxFlexOS/f6cb0c4/DellEMC_VxFlexOS/314 : SUCCESS in 1h 59m 55s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"003d0adf1ded43acfb1e41e930f03a4070e1275b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-13 18:48:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/90/707390/3/upstream-solidfire-check/cinder-solidfire-iSCSI/748258c/ : SUCCESS in 59m 12s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"1d35d3ad84d830f0b442e7daeba6bf15fd4dff79","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-02-13 18:57:26.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/90/707390/3/check/synology-dsm-cinder-iscsi/400203f : UNSTABLE in 1h 06m 57s","accounts_in_message":[],"_revision_number":3},{"id":"402147e5876713ea8e35d14c47ff8bb9f034a7ab","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-13 19:13:33.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/3/check/DellEMC_XtremIO_FC/b094dfe/DellEMC_XtremIO_FC/531 : FAILURE in 48m 18s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/3/check/DellEMC_XtremIO_iSCSI/79b739e/DellEMC_XtremIO_iSCSI/516 : FAILURE in 1h 06m 39s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"d9be6df0428d4ea25adaf1a39a2df998c2ebeb84","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-02-13 19:53:21.000000000","message":"Patch Set 3:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3305 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"d0b67c56fc0fe9f215a9bf36805779ec65c5f17c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-02-13 20:04:05.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/707390-3/Seagate-iSCSI : SUCCESS in 37:03.62\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/707390-3/Seagate-FC : SUCCESS in 33:01.71","accounts_in_message":[],"_revision_number":3},{"id":"f9ec2cd49bf59f589aea84bf120c75b6fc8ab66e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-02-13 20:15:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/707390/3/check/DellEMC_SC_FC/ff30f23/DellEMC_SC_FC/314 : SUCCESS in 1h 08m 07s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/707390/3/check/DellEMC_SC_iSCSI/d0396ec/DellEMC_SC_iSCSI/315 : FAILURE in 1h 44m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"10f8fa9bd69afb29891f212e0e6a48956d361c59","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-02-13 20:44:04.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/707390-3/HPMSA-iSCSI : SUCCESS in 44:50.08\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/707390-3/HPMSA-FC : SUCCESS in 44:47.74","accounts_in_message":[],"_revision_number":3},{"id":"9269d1d01386073b1e35a55e1d0f33d09462b8f2","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-02-13 21:02:10.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/707390-3/Lenovo-iSCSI : SUCCESS in 47:44.42\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/707390-3/Lenovo-FC : SUCCESS in 43:58.42","accounts_in_message":[],"_revision_number":3},{"id":"c0ce763df2ad9fca3fa1177f680ea8f01e466cd9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-13 21:18:17.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ec26e77109fdd32ec1c5b6c9c21f7d3f989b465e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-13 21:47:07.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"354a8103f5dec414482e02c72a2abb401db0a573","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-13 21:54:44.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c4946619f47a75b09547ddfa3946fb91c7505933","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-02-13 22:30:30.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/707390/3/check/DellEMC_VNX_iSCSI/bbc152f/DellEMC_VNX_iSCSI/309 : FAILURE in 2h 55m 17s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/707390/3/check/DellEMC_VNX_FC/8a0fcf2/DellEMC_VNX_FC/310 : SUCCESS in 1h 09m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"5167c1a5d471ed90e77c9da433f49136a694039e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-14 02:37:02.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2090/707390/1/check/infortrend-cinder-volume-iscsi/13246959c8207a895f762448bf483d091c73383e : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"36b11a280d990792e7a1dd56be09fabcf212f534","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-14 02:44:51.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a903ceab2209dd8375172b9b5fe9e7c05ad30d7e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-14 02:46:55.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"65b38a2e3023f5e382d7984a892b0428e9f8772e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-14 03:16:10.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"01cdf726c4b9b32df03a4d5b0203d6b026353d8f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-14 03:20:49.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"74a142ddf009b58545ba7aa9bccd5f31f82fbae2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-14 03:22:42.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2092/707390/1/check/infortrend-cinder-volume-iscsi/13246959c8207a895f762448bf483d091c73383e : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"af4ba8d5afb96647d15312b557960af7e0550406","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-14 03:58:55.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"124f1dc44befd04a8c27270736c72b3720c05c71","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2020-02-14 09:07:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"61564db1012aca33d1e351cac4c0c0c9613933f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-14 10:54:01.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2102/707390/2/check/infortrend-cinder-volume-iscsi/334a9327f91f74b6a98cb00230f45490f38826c2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"268bc84c917387834eb22b58a9e965712bac9817","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-14 11:55:05.000000000","message":"Patch Set 3:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2103/707390/3/check/infortrend-cinder-volume-iscsi/eeadb174980df63ab96f05b6c423842124ff510b : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"839711880996841116016c47d825ba377aee7020","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-02-14 13:29:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/90/707390/3/check/cinder-storpool-tempest/a8b02f7/ : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":3},{"id":"3b56fca797124b35613ead70228ba23d73dd3aa6","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-02-14 15:41:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/3/20200214221844/NEC-FC : SUCCESS in 2h 11m 58s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/3/20200214221844/NEC-ISCSI : SUCCESS in 2h 22m 49s","accounts_in_message":[],"_revision_number":3},{"id":"4f2a42db366fa41be62d203b87e4090c534efe5c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-02-15 04:47:38.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/90/707390/3/check/dsvm-tempest-infinibox-fc/4a8d377 : EXCEPTION\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/90/707390/3/check/dsvm-tempest-infinibox-iscsi/5a0be85 : EXCEPTION","accounts_in_message":[],"_revision_number":3},{"id":"5876f7a1fa79696f06b48ab8de16aefc95b2bb5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-15 07:55:40.000000000","message":"Patch Set 3:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2107/707390/3/check/infortrend-cinder-volume-iscsi/eeadb174980df63ab96f05b6c423842124ff510b : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"fca413ec51e1a4c0e510bc4e662549da980b2f38","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-15 20:09:29.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"71e7bcd0cd861623c5b9f0ec882efc2f8f4e4e83","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-02-18 18:49:13.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nlike test coverage.","accounts_in_message":[],"_revision_number":3},{"id":"a7e7998f9ed8efee5c127626723f3a64df4e5197","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-19 07:58:25.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eec98c63d06a6bce833744568832bdc92719b468","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-19 10:16:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1ebd555acf5c6f031d0fc9fa2d47c81c8452a35d","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-02-19 10:27:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/4/check/nimble-iscsi-driver-dsvm-volume/2cc69ad : FAILURE in 12s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/4/check/nimble-fc-driver-dsvm-volume/949f1f5 : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":4},{"id":"4594e830fa587a835748da13d81c0eac5721d29c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-02-19 10:33:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/90/707390/4/check/cinder-storpool-tempest/1200439/ : FAILURE in 13m 01s","accounts_in_message":[],"_revision_number":4},{"id":"96ef48be8739564e66d990e5b50e73f0b99d4f89","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-19 10:49:25.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/90/707390/4/check-not-vote/ext-cinder-zuul/8576e27 : FAILURE in 32m 08s","accounts_in_message":[],"_revision_number":4},{"id":"9518389ca2f8879da7d6dda66864912832cfe956","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-19 11:10:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/214bd91 : SUCCESS in 53m 56s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/e021b1f : FAILURE in 1m 43s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"5b7a06d6cf7c55b10ff7a9fd612c91faa9ae2e34","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-19 11:25:37.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/4/iscsi/ : FAILURE in 1h 08m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/4/smb/ : FAILURE in 29m 19s","accounts_in_message":[],"_revision_number":4},{"id":"63e97c0f0a8a690e593ea088b4ab01b20c77c04f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-02-19 11:46:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/707390/4/check/dsvm-tempest-zadara-driver/e2d2c58 : SUCCESS in 1h 08m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"da5dbf93785859c2639d4f058e19087ff528d79b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-19 11:59:10.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/4/check/tempest-dsvm-full-bionic/d7a11e1/ : SUCCESS in 1h 29m 39s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/4/check/devstack-plugin-ceph-dsvm-bionic/d139986/ : SUCCESS in 1h 36m 39s","accounts_in_message":[],"_revision_number":4},{"id":"a0ec3137c30a0c92378aff828f425d8a57ffb014","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-02-19 12:02:04.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-4/ns5_nfs : SUCCESS in 1h 43m 56s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-4/ns5_iscsi : SUCCESS in 1h 45m 9s","accounts_in_message":[],"_revision_number":4},{"id":"6a32ca8fea2d31fba90c46b2f6019c41176cc1b9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-19 12:03:03.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/4/check/fujitsu-eternus-dx-iscsi/064d2bd : SUCCESS in 1h 15m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/4/check/fujitsu-eternus-dx-fc/89e8a1f : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":4},{"id":"16ccea7f1703681aa5261415f137cfd571b5489e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:04:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/300b8d8325d64cbca2958963f930f430 : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b2487eb46149f8aa7a6cda4d848b11 : SUCCESS in 10m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d9981386498415b83c507d74886b619 : SUCCESS in 12m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa0aaef41a6c4a88813e2b69c4c0aada : SUCCESS in 12m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2905b41ecd6c4e378fd53b0f44e3a471 : FAILURE in 11m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59a48d0d51540da9190991e81312fe6 : SUCCESS in 17m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eb53e01373fd47acb2bb0f9aa449db65 : SUCCESS in 1h 04m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/904977ec5f26465582e5d9f680c63898 : SUCCESS in 1h 20m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9a34a8c390943ddbe9eefefc33df594 : SUCCESS in 9m 38s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0dce44305d094e3e906e49d2304b7cef : SUCCESS in 1h 26m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f14e0ade7ec44b01b126b7f11afac1b4 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4d680ecd6304526abfde49bc67fb268 : SUCCESS in 7m 12s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2c093c4f5db04ba08fa9c4ad6289144f : SUCCESS in 7m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2c3acd6b89c946d3a85492ebf8aa6668 : SUCCESS in 48m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8aa910d59eb04e408b26902d4acdfd77 : FAILURE in 8m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/29f4beb461ca4561832bdd36b2c9df09 : FAILURE in 21m 16s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dbb73f1ee46b4bd997b3edf3e0f932e6 : SUCCESS in 1h 11m 57s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d6d6f2a0bc624ea391224b963f76c30a : SUCCESS in 1h 19m 53s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/d20be239a9244193bd5c24b610cda632 : SUCCESS in 1h 30m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3009db0ab9046619b001c605d23fb13 : FAILURE in 2h 19m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e608c0502162435f932290da18be9d42 : SUCCESS in 1h 03m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c06484e3bfe6eb273faab35a014706588dac432a","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-19 13:18:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2ca56d50ce9b55e5eac064e4b311a560ce613297","author":{"_account_id":31416,"name":"wuqiongdan","email":"wuqiongdan@szsandstone.com","username":"wuqiongdan"},"date":"2020-02-19 13:24:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- SandStone-iSCSI http://47.75.113.2/90/707390/4/check/SandStone-iSCSI/c314435/ : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":4},{"id":"2fed45207fdfdd9841bd026c1f8497f03f869d54","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-19 13:33:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/707390/4/check/dsvm-tempest-cisco-zonemanager-job/fe01697/ : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":4},{"id":"aeadb61d4532ab0ad6d839ecfa8ed374db70befe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-02-19 13:36:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/90/707390/4/check/cinder-storpool-tempest/dd3326f/ : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":4},{"id":"9fbeff46f113ea02db23e5c6c97d90b9b473a023","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-19 13:45:45.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/90/707390/4/check-not-vote/ext-cinder-zuul/0dcc810 : FAILURE in 26m 10s","accounts_in_message":[],"_revision_number":4},{"id":"8075df5e7a294ff24476a55e7d4725360b737ace","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-02-19 14:01:51.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1694 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1901 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1728 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"1aabaa4ce16bf2c76900f3c0ebd5b49ef702617e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-19 14:21:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/4/iscsi/ : SUCCESS in 1h 02m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/4/smb/ : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":4},{"id":"57f194183560996efd865d7e43a7a474eb56e8f2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-19 14:24:52.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c3d5bb9 : FAILURE in 51s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/8b2e180 : SUCCESS in 1h 05m 38s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"f9718a13ea6d321f0b0779b5a8e4101df7a71cbc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-19 14:27:34.000000000","message":"Patch Set 4:\n\nBuild Successful. \nFor 3rd party ci contact info:\n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2090/707390/4/check/infortrend-cinder-volume-fc/109866f179c3bed989cd1cecce91c5d780dfe66f : SUCCESS\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2132/707390/4/check/infortrend-cinder-volume-iscsi/109866f179c3bed989cd1cecce91c5d780dfe66f : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"6af0318f3e58a80d0fbe331b5f9df3e9e5760949","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-02-19 14:34:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/707390/4/check/dsvm-tempest-zadara-driver/a426b31 : SUCCESS in 1h 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cfd460ac33e933c5cfc1582c04419e185525370b","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-02-19 14:55:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cfecc5398bb85c6141b873b2ce2568c85b684ac9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-19 15:01:38.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/4/check/tempest-dsvm-full-bionic/05aaf81/ : SUCCESS in 1h 34m 19s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/4/check/devstack-plugin-ceph-dsvm-bionic/279d2de/ : SUCCESS in 1h 37m 08s","accounts_in_message":[],"_revision_number":4},{"id":"0538ea588e655f77aee4b071f1e253f97c1d06e9","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-19 15:33:33.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"dcc7e7d879814f039f475b9c292c07552b0833aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 16:04:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62e39275a9a54e5ea5e4ba8714a5ed8a : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c937b287b9564742945ad80d5f334072 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c32bb99b8bb47879273e28104edfe2e : SUCCESS in 12m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19905158a8f8441b8ddc256d263cc117 : SUCCESS in 12m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea86d757479942d3a53f8eadf8f2c317 : FAILURE in 12m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3160ce252c864379baff01533ffc3751 : SUCCESS in 16m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c5a5d819431d49e694453e6c9fa7976f : SUCCESS in 1h 03m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a52650bfbc8e422494554f84f4d429c3 : SUCCESS in 1h 17m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba4d494bde5d4d23a4557ce9fa69bb76 : SUCCESS in 10m 43s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14af666ea10b4fffaeeb8242ca4ed853 : SUCCESS in 1h 32m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f71e1fd342454fdb98429ab645d9edfb : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9951860a1f414a8b96f408d83c2b3689 : SUCCESS in 7m 20s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/feae2a3b53384ffaa7d643c12f6dfeff : SUCCESS in 7m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f6da11e8b7324299bc3c2a78f94221ac : SUCCESS in 46m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/15cc570dec224626abeaa419b893cd0b : FAILURE in 8m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd389160f9e34847b97c3b6bb1f430ee : FAILURE in 15m 23s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/82c08aaa562e44cabf003a4567c25606 : SUCCESS in 1h 19m 52s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1dd37ec2042448048754e2028e65d8df : SUCCESS in 1h 19m 58s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/826af0d9803b4b7589a62265c9ff009b : SUCCESS in 1h 29m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94e1e56902e049da9b904fe7877b9946 : SUCCESS in 2h 33m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3537422924b04467a75929c99a5d9d55 : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":4},{"id":"c8aec36bacfb54675e2665e6021c25c32b51c81f","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-02-19 16:16:24.000000000","message":"Patch Set 4:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3333 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"53d66d05ab11d0c3dd741d9d79e3f8da24eab7e1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-19 16:48:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/4/check/DellEMC_XtremIO_FC/3394826/DellEMC_XtremIO_FC/584 : SUCCESS in 1h 25m 04s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/4/check/DellEMC_XtremIO_iSCSI/67ddc3c/DellEMC_XtremIO_iSCSI/565 : FAILURE in 1h 51m 17s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"f29a90466e2a4e8b62160056b49d41599faf4b12","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-19 16:49:29.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c440e8cae721c076f83006062b1895fa163b6503","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-19 17:06:23.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-90-707390-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"ff55bb0921f80842c1001e84ae2f0321d65ee6a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-19 17:08:51.000000000","message":"Patch Set 4:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2091/707390/4/check/infortrend-cinder-volume-fc/109866f179c3bed989cd1cecce91c5d780dfe66f : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2133/707390/4/check/infortrend-cinder-volume-iscsi/109866f179c3bed989cd1cecce91c5d780dfe66f : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"778966b37a836ad534ce04dbc2fd8139f0a8e968","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-19 17:09:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/90/707390/4/upstream-solidfire-check/cinder-solidfire-iSCSI/852601f/ : SUCCESS in 1h 02m 44s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"dd1ed97bacf5f9e7cafa39f75d4557ea6852a879","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-02-19 17:09:09.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/90/707390/4/check/synology-dsm-cinder-iscsi/6581cb4 : UNSTABLE in 1h 03m 47s","accounts_in_message":[],"_revision_number":4},{"id":"deb74a2bde3cd6243e97226f16b8fddc29c9e2fc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-19 17:26:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/90/707390/4/upstream-check/cinder-cDOT-iSCSI/ab71556/ : FAILURE in 1m 52s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/90/707390/4/upstream-check/cinder-cDOT-NFS/c6a93bd/ : SUCCESS in 54m 23s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/90/707390/4/upstream-check/cinder-cDOT-FCP/8529604/ : FAILURE in 1h 20m 28s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"8c863e70390b74759dc6a789da7a52fdeb7351ab","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-02-19 17:42:37.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/707390-4/Seagate-iSCSI : SUCCESS in 39:27.60\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/707390-4/Seagate-FC : FAILURE in 25:59.29","accounts_in_message":[],"_revision_number":4},{"id":"e2c480cf038e438efc7d5745b25db2ad0ac50263","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-02-19 18:01:31.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/707390-4/Lenovo-iSCSI : SUCCESS in 44:24.80\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/707390-4/Lenovo-FC : SUCCESS in 40:07.71","accounts_in_message":[],"_revision_number":4},{"id":"434448a1a5c886e7655d3e8e2e831c14896c8e65","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-02-19 18:54:48.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/707390/4/check/DellEMC_SC_FC/d636525/DellEMC_SC_FC/333 : FAILURE in 1h 13m 06s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/707390/4/check/DellEMC_SC_iSCSI/fd38225/DellEMC_SC_iSCSI/334 : FAILURE in 1h 53m 39s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"76985141d77be1e608173ba88345cc467a035fb5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-02-19 19:54:35.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/707390/4/check/DellEMC_VNX_iSCSI/a0bef52/DellEMC_VNX_iSCSI/325 : FAILURE in 2h 52m 40s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/707390/4/check/DellEMC_VNX_FC/e07b06c/DellEMC_VNX_FC/327 : SUCCESS in 1h 00m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"36ded9d5afb39fe9eb2b495d9dc6504462247e7d","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-02-19 22:50:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/4/20200220053539/NEC-ISCSI : SUCCESS in 2h 13m 33s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/4/20200220053539/NEC-FC : SUCCESS in 2h 14m 27s","accounts_in_message":[],"_revision_number":4},{"id":"3aa1f530294c4b563fdefa9f75a2cd90950b322d","author":{"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},"date":"2020-02-20 03:35:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/90/707390/4/check/kaminario-dsvm-tempest-full-iscsi/1f4d255 : SUCCESS in 1h 58m 25s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/90/707390/4/check/kaminario-dsvm-tempest-full-FC/514a0d1 : SUCCESS in 1h 55m 30s","accounts_in_message":[],"_revision_number":4},{"id":"258e72f546425fa06f51586ec2efbb2550de5634","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-20 03:43:58.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"5b57a8e5e1acaac0ce2c2466bae240a0a9d24d22","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-20 03:46:18.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a283e1bee696246c7b9c40a313ce8bac7554656b","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-02-20 09:54:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/275/index.html : FAILURE in 1h 13m 07s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/424/index.html : SUCCESS in 1h 35m 28s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/467/index.html : SUCCESS in 1h 23m 57s","accounts_in_message":[],"_revision_number":4},{"id":"eb387c2731719a9a8b59edbf598281d1c88d1232","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-21 04:44:44.000000000","message":"Patch Set 4:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":4},{"id":"8ce441a2629f506775c2c70a813b943879054343","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-24 16:15:05.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nA minor nit and a question, but we also still need third party CI results.","accounts_in_message":[],"_revision_number":4},{"id":"9d4d0e51613831b4324a2b37552b73a79153d047","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-02-24 20:34:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nNeed vxflexOS CI result for the patch.\nDo you need to do something special for volume delete and for cg delete? expect that volume does not have a replication pair before it can be deleted. Ditto for cg, cannot be in use by any replication pairs.","accounts_in_message":[],"_revision_number":4},{"id":"f3cbdc2a6a130424625473fe26f11587fb2ee392","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-25 09:15:45.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e (2 comments)\n \u003e \n \u003e Need vxflexOS CI result for the patch.\n \u003e Do you need to do something special for volume delete and for cg\n \u003e delete? expect that volume does not have a replication pair before\n \u003e it can be deleted. Ditto for cg, cannot be in use by any\n \u003e replication pairs.\n\nCG is not deleted by driver. Storage administrator creates VxFlex OS CG and pass it\u0027s name to the driver.\nDuring volume deletion we firstly delete volumes pair, only after that replicating volume and replicated volumes can be deleted.","accounts_in_message":[],"_revision_number":4},{"id":"476f40fd42eb6c957196e954fd0e2c0eef2e8c52","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-25 09:23:47.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2fed557eac932f5f1f48ea69a63104cd2b1a846a","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-25 09:24:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8296195f11ab046f2b4f5f72bbe0d09d6bd96e59","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-25 10:27:22.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/5/iscsi/ : FAILURE in 1h 00m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/5/smb/ : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":5},{"id":"2f21cc5d6f1bab1ddcbd2118d10f218c5c7a08c8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-25 11:10:14.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/5/check/tempest-dsvm-full-bionic/99a79c5/ : SUCCESS in 1h 32m 33s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/5/check/devstack-plugin-ceph-dsvm-bionic/9318fc7/ : SUCCESS in 1h 40m 44s","accounts_in_message":[],"_revision_number":5},{"id":"7ccdec59ee8e174cfd8772b1ed6a9651826bafd8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-25 11:24:21.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/5/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/dea94c3 : FAILURE in 1h 59m 14s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/5/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/7d844b1 : FAILURE in 17m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"01c4efd4b84aaf4e06edd921babbfa157bf19447","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-02-25 11:35:17.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-5/ns5_nfs : SUCCESS in 2h 10m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-5/ns5_iscsi : SUCCESS in 1h 53m 3s","accounts_in_message":[],"_revision_number":5},{"id":"38b4aec4395e1ac5a38d8c5935a071d0dd8d0abf","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-25 11:58:18.000000000","message":"Patch Set 5:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/5/check/fujitsu-eternus-dx-iscsi/5144612 : SUCCESS in 1h 21m 33s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/5/check/fujitsu-eternus-dx-fc/15fe354 : SUCCESS in 1h 26m 07s","accounts_in_message":[],"_revision_number":5},{"id":"e36f5a4e801ad53559e800ac6ca167ebbae0d08d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-25 12:59:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/707390/5/check/dsvm-tempest-cisco-zonemanager-job/6049571/ : SUCCESS in 1h 05m 19s","accounts_in_message":[],"_revision_number":5},{"id":"5745d170f0cefb62e0be292752872cc7c71fa12a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:04:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93fe8484d1354d1c95766c128ea0c700 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d654ed7078f42f5a6582519f7e7426c : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72ebfc1869124715b31de68b4c458553 : SUCCESS in 13m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39c21a081be44ffab8eaa62dca1ab6d7 : SUCCESS in 16m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1190d6d943c741d395694cd22fbfe36f : FAILURE in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5e0064bda34b65a2ca8770b708ac5c : SUCCESS in 19m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f773d285f7dd41ebaf87de9687e705e4 : SUCCESS in 57m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fe15faba05645f5839716d09998d38b : SUCCESS in 1h 34m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52cbc6d7cabe4e1baa11c70dbcda2ba8 : SUCCESS in 20m 03s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6cf09dc32b54435482eb035dc8b890d8 : SUCCESS in 1h 19m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d9ec728a7cc94d36a7fcdebf0358801d : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/581c7dde844748ec8cad646d78939702 : SUCCESS in 10m 32s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/82293621add94676b58c617fba1cec79 : SUCCESS in 15m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0997920c9de4499b89c56fdad9ceafa7 : SUCCESS in 43m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89dd9f746b974f8e96a5a58af706353b : FAILURE in 11m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7e2caea36cf94170810047248464fb3c : FAILURE in 23m 10s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9f40986469744e358326680740c40f2d : SUCCESS in 1h 18m 33s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7d1447a95edf439da58f13dce005aa00 : SUCCESS in 1h 20m 55s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/27aa8655000242c2b2adc7cb915ba3c2 : SUCCESS in 1h 28m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54d60a504d6d4dd9b3e60b355c5ae594 : SUCCESS in 2h 43m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16c3340814d547e081f7d921a8808cfa : SUCCESS in 1h 15m 02s","accounts_in_message":[],"_revision_number":5},{"id":"e14ede9c35340975ff0e68bda5e24e6a0ee6c998","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-25 13:49:04.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"8aa5a548ccc80386d6129f1a7ae6b957575342e8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-25 13:50:40.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/5/check/DellEMC_XtremIO_FC/28bf29d/DellEMC_XtremIO_FC/620 : FAILURE in 11m 36s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/5/check/DellEMC_XtremIO_iSCSI/7dccb8f/DellEMC_XtremIO_iSCSI/598 : FAILURE in 2h 16m 17s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"f8c415ebd793f8099107e9c452792b7afc155d8f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-25 14:07:09.000000000","message":"Patch Set 5:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-90-707390-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"b5db924af68a5883d5125a3804bf1d698493ae0f","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-02-25 14:15:23.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/90/707390/5/check/synology-dsm-cinder-iscsi/08caaf7 : UNSTABLE in 1h 09m 50s","accounts_in_message":[],"_revision_number":5},{"id":"89008a990edc46a6e264020fd642c7ee8242a0ad","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-02-25 14:40:38.000000000","message":"Patch Set 5:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3361 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"db8acde78103b4adba2bb4bbf4325d9290206a65","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-25 14:57:09.000000000","message":"Patch Set 5:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":5},{"id":"70984148454920e95a026b8ca065a799440e4e2a","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-02-25 15:09:08.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/707390-5/Lenovo-iSCSI : SUCCESS in 46:45.10\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/707390-5/Lenovo-FC : SUCCESS in 44:20.76","accounts_in_message":[],"_revision_number":5},{"id":"a62d7300dc52b2a898c8caf01884a7f054eb5f34","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-02-25 15:10:35.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/5/check/DellEMC_VxFlexOS/9db0374/DellEMC_VxFlexOS/332 : SUCCESS in 2h 05m 21s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"1d559b0699d06f8319992268034b1d92ca16e450","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-25 15:23:00.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"14da3922114d0773deb555c819e9915377c4ae14","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-02-25 15:49:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/707390/5/check/DellEMC_VNX_iSCSI/6053aae/DellEMC_VNX_iSCSI/345 : FAILURE in 2h 44m 45s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/707390/5/check/DellEMC_VNX_FC/ee1235a/DellEMC_VNX_FC/347 : SUCCESS in 1h 00m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"a35cbc02b46faf30ce4fe7148ca0e2d9bf7e5a39","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-25 17:02:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/90/707390/5/upstream-check/cinder-cDOT-iSCSI/fe167cc/ : SUCCESS in 57m 17s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/90/707390/5/upstream-check/cinder-cDOT-NFS/718972a/ : FAILURE in 3h 56m 14s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/90/707390/5/upstream-check/cinder-cDOT-FCP/73dc705/ : SUCCESS in 1h 40m 36s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"07e3a301daca7dff80aa7931edc83a6e9fd6f91d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-25 17:04:49.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/90/707390/5/upstream-solidfire-check/cinder-solidfire-iSCSI/54b1ad8/ : FAILURE in 3h 58m 20s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"8b0d6680c19a9f3f922e7e3ad8e0e1cdd06e8162","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-02-25 18:04:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/5/20200226003916/NEC-FC : SUCCESS in 2h 20m 03s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/5/20200226003916/NEC-ISCSI : SUCCESS in 2h 25m 15s","accounts_in_message":[],"_revision_number":5},{"id":"01999ae79390feefba8dc9854d232b772cf5a1cf","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-02-25 18:35:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/707390/5/check/dsvm-tempest-zadara-driver/61cf183 : SUCCESS in 1h 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8a9a11d33454f3ff4b2b51942c23ee498c036afb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-25 19:20:00.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"99a6e03a525d459890b37b4ac9adb8dfb7dec8a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-25 21:01:13.000000000","message":"Patch Set 5:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2158/707390/5/check/infortrend-cinder-volume-fc/00122a9cbb65e037f783505dc11acfaec975743c : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2200/707390/5/check/infortrend-cinder-volume-iscsi/00122a9cbb65e037f783505dc11acfaec975743c : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"c4a9b6030a79efb6aeb177af21a9d6831fc03083","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-02-25 21:27:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/326/index.html : SUCCESS in 1h 09m 53s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/473/index.html : SUCCESS in 1h 36m 47s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/516/index.html : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":5},{"id":"1b51deb8a9f416b683880241a8ae0ff739363143","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-02-27 02:12:59.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/90/707390/5/check/dsvm-tempest-infinibox-fc/aec8341 : FAILURE in 19m 08s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/90/707390/5/check/dsvm-tempest-infinibox-iscsi/0733179 : FAILURE in 1h 03m 16s","accounts_in_message":[],"_revision_number":5},{"id":"1ae06eaf524a934890fd7d54ba4d31372bdf843a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-27 13:27:18.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"e1edd00d3cdc9474d72a0390904f6e6ef0824a0b","author":{"_account_id":22495,"name":"Kaminario K2FC CI","email":"cinder.fc.devops.CTR@kaminario.com","username":"k2-fc"},"date":"2020-02-28 02:55:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/90/707390/5/check/kaminario-dsvm-tempest-full-iscsi/62b1493 : SUCCESS in 1h 59m 09s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/90/707390/5/check/kaminario-dsvm-tempest-full-FC/45e213f : SUCCESS in 2h 20m 03s","accounts_in_message":[],"_revision_number":5},{"id":"2dadc611dd6d77f36314347188fc679b97bf5ed7","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-28 10:17:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"19923554ee4c0a7cf05e6e8328443d4567bb7bd2","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2020-02-28 10:25:26.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"51323b3e968b1383c754e9cb2a52c2f8d3a241db","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-02-28 10:30:00.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1713 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1741 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1915 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"c45344a32096d09f719ac71df50a5d4c92b42fa7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-02-28 11:31:10.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/6/iscsi/ : FAILURE in 55m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/6/smb/ : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":6},{"id":"a66c50779e9f3498833ba6c2aedbf2a39becaafa","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-28 11:31:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3b9f1f8 : FAILURE in 23m 44s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/eb67bc2 : UNSTABLE in 1h 12m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"d32be0728bdd25c933f0ed91bd40ebfe20c3df79","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-02-28 12:06:36.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-6/ns5_nfs : SUCCESS in 1h 37m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-6/ns5_iscsi : SUCCESS in 1h 48m 18s","accounts_in_message":[],"_revision_number":6},{"id":"da334830623b626d3f9a351cddaf7a14cf95e0b0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-02-28 12:28:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/707390/6/check/dsvm-tempest-cisco-zonemanager-job/9739548/ : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":6},{"id":"3d905972147b5f8cb4af9f96d1210ad257cad769","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-02-28 12:38:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/707390/6/check/dsvm-tempest-zadara-driver/2b0b64c : SUCCESS in 1h 09m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7ea2f460128d373851bf1ac61b4478185a02d90f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-28 12:58:12.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/6/check/tempest-dsvm-full-bionic/468cf49/ : SUCCESS in 1h 26m 05s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/6/check/devstack-plugin-ceph-dsvm-bionic/2f4a0b3/ : SUCCESS in 1h 40m 27s","accounts_in_message":[],"_revision_number":6},{"id":"05d587f7040f1f73b5c6b9ffa0e7142e6f3329b2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-02-28 13:06:55.000000000","message":"Patch Set 6:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2200/707390/6/check/infortrend-cinder-volume-fc/bf1088ef32335c2f7d2a9f5a9e76cd527119074d : FAILURE\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2242/707390/6/check/infortrend-cinder-volume-iscsi/bf1088ef32335c2f7d2a9f5a9e76cd527119074d : UNSTABLE","accounts_in_message":[],"_revision_number":6},{"id":"f53760bffb0e08fce04ffebb35e76f13f439c033","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-02-28 13:23:23.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/6/check/fujitsu-eternus-dx-iscsi/112cab8 : SUCCESS in 1h 26m 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/6/check/fujitsu-eternus-dx-fc/dae98e8 : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":6},{"id":"716ae4e457f26039e65169699f19010113977c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 13:42:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b399d40f287746deb9bca5958c7e567c : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2922cb385e484e1e8525f86f628aa0b9 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c25e4e8da0da40c2a45d0bba03b04dd5 : SUCCESS in 10m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8dafff2d01c0473a804af8d60bd2288b : SUCCESS in 14m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70d9b059042847d2823bbc72cacc0539 : FAILURE in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6273aec54a74ab2a17ba02f0e6fef54 : SUCCESS in 16m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/67345860b7db4e1ea67ca46c15821b13 : SUCCESS in 1h 07m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99807c83c5f24106bfadf85e2df3998d : SUCCESS in 1h 17m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71a3194b311c4613ab95589c69407df2 : SUCCESS in 18m 45s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1291cbd905b447b98a09b4a3fe64f144 : SUCCESS in 1h 29m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77abba4b148f4cc5b724de1001bd2442 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d82f4dc4c084082834a23fcc9dcd128 : SUCCESS in 7m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/639a5e0756274a7daf80823b9f98efaa : SUCCESS in 8m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a6e4359c85604741a515eeb311a88288 : SUCCESS in 42m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/919c725a0eaf49dabdd926aaa7d384f2 : FAILURE in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0b4b813ad2664fccaf4f7f4de8efa3b5 : SUCCESS in 1h 03m 49s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc2d4213e90a435e85ac36caf49b9747 : SUCCESS in 1h 10m 00s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82e15524642b4b63b4c00ce203f26d54 : SUCCESS in 1h 29m 03s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/fbdcf10f382b4bbcb12be677dc04b9b4 : SUCCESS in 1h 36m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9fb03d31cd7446e5b3c7a5552c509c57 : SUCCESS in 2h 40m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2325e47b16884433a496426afe540c59 : SUCCESS in 1h 13m 01s","accounts_in_message":[],"_revision_number":6},{"id":"6f621e01c1d8d5e76e225e0ac73465f61ec4ba5a","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-28 14:08:25.000000000","message":"Patch Set 6:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":6},{"id":"66217cb6529f89af836cfae98b03288ab5e668d2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-28 14:36:48.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/90/707390/6/upstream-solidfire-check/cinder-solidfire-iSCSI/9cfc376/ : FAILURE in 4m 47s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"0a589dce430dbd06a87df21d06ded33d55c5c682","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-28 14:47:37.000000000","message":"Patch Set 6:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-90-707390-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6},{"id":"0ad4516643f6bb7dc57b1035e1e1b1bfc06e5051","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-02-28 15:15:10.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/90/707390/6/check/synology-dsm-cinder-iscsi/980efa1 : UNSTABLE in 1h 11m 07s","accounts_in_message":[],"_revision_number":6},{"id":"9ff43479718a960a67ffcb39cb8e1e3f48a557ba","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-02-28 15:26:29.000000000","message":"Patch Set 6:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3383 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"1115de1b4b885d82d6bb05ad3b351a62de7b0a0e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-28 15:27:11.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/90/707390/6/upstream-check/cinder-cDOT-iSCSI/cf461fd/ : SUCCESS in 45m 19s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/90/707390/6/upstream-check/cinder-cDOT-NFS/3403a13/ : FAILURE in 1h 43m 40s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/90/707390/6/upstream-check/cinder-cDOT-FCP/78f8965/ : SUCCESS in 59m 45s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"cce9e9ff484b7f236075c9647f4a3b9e80944e0a","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-02-28 15:39:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/349/index.html : SUCCESS in 1h 07m 54s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/496/index.html : SUCCESS in 1h 35m 17s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/539/index.html : SUCCESS in 1h 23m 40s","accounts_in_message":[],"_revision_number":6},{"id":"c7c6bca9cbb68574726d9605a917c8483f479dab","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-02-28 18:15:38.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/707390/6/check/DellEMC_SC_FC/1c6efaf/DellEMC_SC_FC/372 : SUCCESS in 1h 01m 19s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/707390/6/check/DellEMC_SC_iSCSI/9be8f43/DellEMC_SC_iSCSI/373 : SUCCESS in 1h 33m 25s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"361119556334aa0c4b105c445de4ac5d9ebc9a29","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-02-28 19:26:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"0866dc3321ba4b0450d351c200694b42470234fa","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-02-28 19:33:26.000000000","message":"Patch Set 6:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":6},{"id":"c731e78f5ecc50be72af638660e40b6ba4fd9adb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-02-28 22:24:06.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/6/check/DellEMC_VxFlexOS/405dded/DellEMC_VxFlexOS/354 : SUCCESS in 2h 09m 24s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"aa6803e68d576e7b3083b5f89035934b1b5a119f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-02-28 23:26:42.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/707390/6/check/DellEMC_Unity_FC/035160a/DellEMC_Unity_FC/540 : SUCCESS in 1h 16m 15s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/707390/6/check/DellEMC_Unity_iSCSI/0f66aa6/DellEMC_Unity_iSCSI/531 : SUCCESS in 1h 41m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"f579bd5c60f411c6c2cc0d5aa57ceb3f3238ddd1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-29 00:29:33.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"2aa73e5d99d7ae2d7d65a6838755d9bfa6351335","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-02-29 00:42:33.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"87a01cd05cee2224bb219d6ebf79f06765540a1f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-02-29 01:02:04.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/6/20200229074207/NEC-FC : SUCCESS in 2h 06m 06s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/6/20200229074207/NEC-ISCSI : SUCCESS in 2h 19m 54s","accounts_in_message":[],"_revision_number":6},{"id":"b58b82aa8cad3f2da09e1086631f80898372c98a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-02-29 02:18:39.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"ce1f642214478b2c431cb1cb642963407d7934ec","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-01 04:52:16.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/707390/6/check/DellEMC_PowerMAX_FC/9b9b30f/DellEMC_PowerMAX_FC/419 : FAILURE in 6h 03m 28s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/707390/6/check/DellEMC_PowerMAX_iSCSI/fb2a738/DellEMC_PowerMAX_iSCSI/349 : SUCCESS in 8h 03m 42s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"e98fbe68f6183c0613ff281514599db7da45d87d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-01 06:40:37.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"c496bfd8a8f650318a3c63991c2d3a32f6b8e98f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-01 08:18:26.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/6/check/DellEMC_XtremIO_FC/db7397f/DellEMC_XtremIO_FC/660 : SUCCESS in 1h 26m 39s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/6/check/DellEMC_XtremIO_iSCSI/a8a1d4f/DellEMC_XtremIO_iSCSI/628 : FAILURE in 4h 46m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"0327cda16b3a18a64608778983a482d4ceb203ca","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-01 08:26:49.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"e5292abf6f866d29320c2b4ec2e74fcca10e0508","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-03-01 09:51:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/90/707390/6/check/cinder-storpool-tempest/87c3bf8/ : FAILURE in 14m 34s","accounts_in_message":[],"_revision_number":6},{"id":"b949f79d9475d1d1267e2b4b1180eeb9bc1e574f","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2020-03-01 20:42:23.000000000","message":"Patch Set 6:\n\nrun-storpoolci","accounts_in_message":[],"_revision_number":6},{"id":"f29ee108e3711e54ff82751d28daf6657363ef32","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-03 06:24:40.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/90/707390/6/check/dsvm-tempest-infinibox-fc/47f70be : FAILURE in 8m 51s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/90/707390/6/check/dsvm-tempest-infinibox-iscsi/ebba451 : FAILURE in 1h 01m 37s","accounts_in_message":[],"_revision_number":6},{"id":"56ae67fb69ea9ba6f953b4e9e92ff4340acc20bf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-03 13:29:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dd0389b0e17068456a1adaa011a8aa4f6130f41f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-03 13:33:38.000000000","message":"Patch Set 6: Code-Review+2\n\nFiled https://bugs.launchpad.net/cinder/+bug/1865879 to track the model updates to not hold this review up. Other than that, code looks fine and CI is happy.","accounts_in_message":[],"_revision_number":6},{"id":"6fb9b1b1bb797bde1e2efa3ff08555212d088a85","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-03-03 14:29:04.000000000","message":"Patch Set 6: Code-Review-1\n\nPlease, update\nhttps://docs.openstack.org/cinder/latest/reference/support-matrix.html\nfor volume replication support.","accounts_in_message":[],"_revision_number":6},{"id":"51ed55f3af3c416865615e03bee948b82266e63b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-03 14:35:16.000000000","message":"Removed Code-Review+2 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":6},{"id":"9aea93088b1f329e35211d9b37a80cd869a8fa51","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-03-03 16:35:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/90/707390/6/check/cinder-storpool-tempest/22e6b56/ : POST_FAILURE in 3h 01m 59s","accounts_in_message":[],"_revision_number":6},{"id":"b5c3959a33b821d8a4bcaafc7762b3978ada02c3","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-04 08:45:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4ecc0ac2b7b4cb1a324214f92b030e37435b7903","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-04 08:46:27.000000000","message":"Patch Set 7:\n\n\u003e Please, update\n \u003e https://docs.openstack.org/cinder/latest/reference/support-matrix.html\n \u003e for volume replication support.\n\nUpdated","accounts_in_message":[],"_revision_number":7},{"id":"9bd74b9d8768c0408e4599251ab80c785c9b03b8","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-03-04 08:57:05.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1723 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1896 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1688 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"a1a865f669bd5a3a837d8b2a8d517b2cbf94c8ba","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2020-03-04 08:59:31.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"448574f163e736ffae8e81ceaff1c0026d0f3f98","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-04 09:21:21.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/5c42afc : FAILURE in 22m 40s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/44c0e2b : FAILURE in 35m 35s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"133e17502beaf163a9e61721973978382575a743","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-04 09:49:45.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/90/707390/7/check/dsvm-tempest-infinibox-fc/f945013 : UNSTABLE in 59m 07s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/90/707390/7/check/dsvm-tempest-infinibox-iscsi/ba07297 : FAILURE in 1h 03m 37s","accounts_in_message":[],"_revision_number":7},{"id":"d8f7007ecf6cc6f4046a0652fc7f19f94563f22a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-04 09:53:45.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/7/iscsi/ : SUCCESS in 1h 07m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/7/smb/ : FAILURE in 1h 03m 28s","accounts_in_message":[],"_revision_number":7},{"id":"f1f6f2d85910dc7ffc86cd9dadd4da5b1ea92ce4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-03-04 09:56:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/707390/7/check/dsvm-tempest-cisco-zonemanager-job/f50ed29/ : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":7},{"id":"95584562db2d2b33d620fafe322640fcf03c2829","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-03-04 10:06:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/707390/7/check/dsvm-tempest-zadara-driver/1275bcf : SUCCESS in 1h 09m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a24dc6e551a4e431edf77cf245e329f24988a5ba","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-04 10:11:43.000000000","message":"Patch Set 7:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/7/check/fujitsu-eternus-dx-iscsi/628b035 : SUCCESS in 1h 25m 26s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/7/check/fujitsu-eternus-dx-fc/2157e83 : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":7},{"id":"8511acc40f9f87fa828bbbbe9d31a055bfbb0c8b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-04 10:15:33.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/7/check/tempest-dsvm-full-bionic/aaa5b0b/ : SUCCESS in 1h 26m 42s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/7/check/devstack-plugin-ceph-dsvm-bionic/efa928a/ : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":7},{"id":"f2e2091652fb0069897d4bcfdde0d021564024db","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-04 10:41:20.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"df801b2b49afe5bdfe6c4be93b3e61192520641f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-04 10:41:32.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-7/ns5_nfs : SUCCESS in 1h 55m 49s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-7/ns5_iscsi : SUCCESS in 1h 39m 26s","accounts_in_message":[],"_revision_number":7},{"id":"0769b1665f238ee67c2fb84203d9a6f4e158e310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:04:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be3ba816339a44dba0d5077f606dc5b7 : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40832014d444fc4bc33d7aa3a14e8ca : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3519414763444348b57b8b80a3fd1f5a : SUCCESS in 9m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbf146cfa4a34f45b13e40315a782a4a : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b72eab8590b5483f80bb8e41973bdafe : FAILURE in 11m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15af6ffa3cdf4318a60176809d5fdd45 : SUCCESS in 16m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d0fbf653a67a48bc91030ec2b34b230e : SUCCESS in 1h 03m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c374552876cd4e5e918a706d523bd64e : SUCCESS in 1h 24m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87e5ed7973974ed2b8f1c3a5d6e43310 : SUCCESS in 16m 47s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/09cd4df76a7b465d874eafb8104d0570 : SUCCESS in 1h 27m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5f987ec026047d895eabfabe7835cbe : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8c89f046f85643e4a347d1bf897782d4 : SUCCESS in 6m 28s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/66ddeeb1ef1d40d195a19b6d26e16bd1 : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c049903f788d416fb57f66eeca963f00 : SUCCESS in 46m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8e4e897bc4a403591dfa7f844b28ad9 : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d90ae837d6f745a1adcd51dcbc0d061f : SUCCESS in 1h 02m 31s (non-voting)\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6304441dad754fe3a9d62d73d5919294 : SUCCESS in 1h 15m 29s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ce9606c9256a43fbbff20aaec219d53f : SUCCESS in 1h 14m 30s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/a3dfda831ec8481a9a868e3cc346acfe : SUCCESS in 1h 28m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6ca4510fa1d4362bf99605eb5d9d0ee : SUCCESS in 2h 17m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84c545b7b4354767be7e70f8408cbc4c : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":7},{"id":"6de9c5c6395831a47484aad6bf58a0a702b29799","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-04 11:16:22.000000000","message":"Patch Set 7:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3408 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"b2d70d071fe48fef8a56c9eff3af1b54a789a453","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-04 11:56:39.000000000","message":"Patch Set 7:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":7},{"id":"72d345457a89cc5d5e8461f418bf2fdc0dc15cfa","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-04 12:14:12.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/90/707390/7/check/synology-dsm-cinder-iscsi/05fd95c : UNSTABLE in 1h 09m 00s","accounts_in_message":[],"_revision_number":7},{"id":"6169798c4052281ba5755db99d13a455280fa7fa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-04 12:15:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/90/707390/7/upstream-check/cinder-cDOT-iSCSI/230f7ad/ : SUCCESS in 53m 44s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/90/707390/7/upstream-check/cinder-cDOT-NFS/6ae4839/ : SUCCESS in 52m 52s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/90/707390/7/upstream-check/cinder-cDOT-FCP/1ed3fbe/ : FAILURE in 1h 09m 17s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"0bd1179a1a49ed29981ae3ee32eb6aa6a54a99cd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-04 12:29:05.000000000","message":"Patch Set 7:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-90-707390-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"64a25ec1ad1c205850225ae81fb8f36ac3f6786b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-04 12:38:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/90/707390/7/upstream-solidfire-check/cinder-solidfire-iSCSI/0180418/ : SUCCESS in 1h 31m 13s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"111a223df389f08da9ae773feed3be055409d7e0","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-03-04 12:49:30.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/707390/7/check/DellEMC_SC_FC/587b3c0/DellEMC_SC_FC/397 : SUCCESS in 1h 08m 50s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/707390/7/check/DellEMC_SC_iSCSI/ef317d7/DellEMC_SC_iSCSI/398 : FAILURE in 1h 44m 29s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"888ed3c9f89034fe3cbde0885cfdcc6b0eb23c98","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 15:45:45.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/7/check/DellEMC_VxFlexOS/451b3a4/DellEMC_VxFlexOS/378 : FAILURE in 2h 10m 58s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"a7f3c6ad9fe3811595023e5b6da81d6721140a57","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-03-04 15:54:34.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"be914d747c8d0bce0017a38bfb5dd33a02504259","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-03-04 16:19:00.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"897815b7b978acdd5697b44fc1f7dc4bc5699d9c","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-04 16:27:39.000000000","message":"Patch Set 7:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":7},{"id":"e50ce6ddd4994f1bed3681367474cdd5e93c2171","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 18:13:49.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/707390/7/check/DellEMC_VNX_iSCSI/91b6277/DellEMC_VNX_iSCSI/375 : FAILURE in 2h 52m 33s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/707390/7/check/DellEMC_VNX_FC/b7f6841/DellEMC_VNX_FC/378 : SUCCESS in 1h 07m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"92a93e46779304f78d62bee1d530a4fa2c5df97a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 18:39:39.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/7/check/DellEMC_VxFlexOS/fb10442/DellEMC_VxFlexOS/379 : FAILURE in 2h 11m 36s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"927ce6ab3b3c12059d0e2a40df36beef4454e2b5","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-03-04 19:26:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/390/index.html : SUCCESS in 1h 11m 14s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/537/index.html : SUCCESS in 1h 37m 19s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/578/index.html : SUCCESS in 1h 26m 03s","accounts_in_message":[],"_revision_number":7},{"id":"78589bdba9e69c6a9ae017a58d5eff8e38591cc8","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-04 19:37:09.000000000","message":"Patch Set 7:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":7},{"id":"2110e38315aceec9eee83a47f4c372d1ed32578b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 23:00:19.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/7/check/DellEMC_VxFlexOS/bc23a73/DellEMC_VxFlexOS/381 : FAILURE in 2h 08m 14s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"2b2ddd0be81b696a4df696f7ab85cce54b7e1ac9","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-03-05 00:39:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/7/20200305072741/NEC-FC : SUCCESS in 2h 10m 30s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/7/20200305072741/NEC-ISCSI : SUCCESS in 2h 11m 25s","accounts_in_message":[],"_revision_number":7},{"id":"21c87f5176c5dddafd85cabae230299534318448","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-05 07:12:57.000000000","message":"Patch Set 7:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":7},{"id":"b1ce0ac1da1b588e881122ceb953b2acfa35a917","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-05 09:22:52.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/7/check/DellEMC_VxFlexOS/ab537ef/DellEMC_VxFlexOS/382 : FAILURE in 2h 09m 32s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"09e1c370741d76c70e980780b66edef52b507180","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-05 13:14:43.000000000","message":"Patch Set 7:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-cinder-volume-iscsi http://openstack.infortrend.com/2286/707390/7/check/infortrend-cinder-volume-iscsi/2ebcb827a58d8174fd02d66baab5ebbf4412fde7 : FAILURE\n\n- infortrend-cinder-volume-fc http://openstack.infortrend.com/2244/707390/7/check/infortrend-cinder-volume-fc/2ebcb827a58d8174fd02d66baab5ebbf4412fde7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"a96470cc2463db7288ae2cf9ec8e8449776b49d7","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-05 13:43:48.000000000","message":"Patch Set 7:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":7},{"id":"45cdfb0e435fcb50d975189bd0bb5d2e8c6bc4e6","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-03-05 13:54:22.000000000","message":"Patch Set 7:\n\nNeed to get DellEMC VxFlexOS CI passing.","accounts_in_message":[],"_revision_number":7},{"id":"f234c7e464c03700aa088d0a66ef5d194ba19fa4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-05 15:55:01.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/7/check/DellEMC_VxFlexOS/f53a107/DellEMC_VxFlexOS/384 : FAILURE in 2h 10m 49s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"4f6780e1481c03a8ad03c2ea01fb339776bb65dc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-06 07:58:46.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/7/check/DellEMC_XtremIO_FC/7bb2982/DellEMC_XtremIO_FC/698 : SUCCESS in 1h 31m 27s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/7/check/DellEMC_XtremIO_iSCSI/3553807/DellEMC_XtremIO_iSCSI/658 : FAILURE in 2h 20m 31s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"53c5f9fa14c4b20aa4bccd4cf7d899991c67048d","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-06 08:48:32.000000000","message":"Patch Set 7:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":7},{"id":"1b050b8cfe5f90a47d2a4c7f11c2effef3ffa85f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-06 10:20:26.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"4e00551ebd07d3ad52164edc66d930f6d6f98090","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-06 10:59:33.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/7/check/DellEMC_VxFlexOS/d17ff26/DellEMC_VxFlexOS/391 : SUCCESS in 2h 10m 37s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"04366a00dea961243be0539ed43a15f253af0b4f","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-03-06 13:27:15.000000000","message":"Patch Set 7: Code-Review+1\n\nGod. CI is passing now.","accounts_in_message":[],"_revision_number":7},{"id":"56bd5cdfce87c1fce9a23b7247e6660121758a1b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-06 13:37:17.000000000","message":"Patch Set 7: Code-Review+2\n\nLooks good - thanks!","accounts_in_message":[],"_revision_number":7},{"id":"87edb6e844c6e471231f4f6495e983662322ce68","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-03-06 13:55:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/90/707390/7/check/cinder-storpool-tempest/3e07f95/ : SUCCESS in 3h 08m 39s","accounts_in_message":[],"_revision_number":7},{"id":"d86f631ba76074873243facf4e3f2b504f2f06ad","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-06 17:05:03.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"86ed1a2806b44ea3ed765ecd434505eae823da42","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-03-09 09:27:31.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"9530e2850bb7ecdfad7b468f8c3d77113cf87c16","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2020-03-19 12:34:22.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b552d72a1af7a39580d9e474d8e57b1ae06cb592","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-03-20 19:25:33.000000000","message":"Patch Set 7:\n\n@Hemna https://review.opendev.org/#/c/705176 is related change and has to be merged before I think","accounts_in_message":[],"_revision_number":7},{"id":"751696c3d3b6ddfc9870906b64b57bf6b625060d","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-03-21 20:22:38.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/90/707390/7/check/kaminario-dsvm-tempest-full-iscsi/eb41d96 : FAILURE in 2h 56m 08s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/90/707390/7/check/kaminario-dsvm-tempest-full-FC/890f2ae : FAILURE in 3h 11m 51s","accounts_in_message":[],"_revision_number":7},{"id":"d48e8723728f8c36604e5c81d9730719c45b6f18","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-31 05:42:41.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"1fbfdc9a475c07df32f7273b5d52ba83d662a397","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-03-31 05:48:13.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"faafbc6e6e422a3a74d77cf506039db2f0b7de9a","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-03-31 06:12:55.000000000","message":"Topic changed from bp/vxflexos-revert-to-snapshot to bp/vxflexos-replication-support","accounts_in_message":[],"_revision_number":8},{"id":"474eee5990aa237358c3687b1d22bd28ba5465ee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-03-31 06:24:23.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/90/707390/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/89b0bf5/89b0bf5.html : FAILURE in 39m 29s","accounts_in_message":[],"_revision_number":8},{"id":"39d77ffaf9af9d9f19ebd6a19bc1a60b5cca2fa7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-31 06:44:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/8/iscsi/ : SUCCESS in 59m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/8/smb/ : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":8},{"id":"7414630b8bb2b3cec2b800a0709fdf2fa2684d06","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-31 07:21:41.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/8/check/fujitsu-eternus-dx-iscsi/05abd2f : FAILURE in 41m 35s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/8/check/fujitsu-eternus-dx-fc/63dd45f : FAILURE in 53m 43s","accounts_in_message":[],"_revision_number":8},{"id":"642bd8da54bbb5e8384d15c0a1bfe3e5a211c1c9","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-31 07:47:21.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/6abc1e9 : UNSTABLE in 1h 02m 13s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/99527a8 : UNSTABLE in 1h 17m 46s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"b306e264556633eaa867af112eff85b633264113","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-31 07:52:04.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-8/ns5_nfs : SUCCESS in 2h 9m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-8/ns5_iscsi : SUCCESS in 1h 50m 54s","accounts_in_message":[],"_revision_number":8},{"id":"f85caf1563e379bdcc273166a5e38a9c8a55dd91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 07:58:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bda8ef44e9343c5ba94cb7f1e0c75fd : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef16d04334547b283d62a87af1d96ed : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c381dc2cc21483abde4d2ef9d3033c0 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a80b918db4754852bd227f7e22f535aa : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dedd708e705e4ce4a68dbb0316edbf96 : FAILURE in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e83bfaf2934d0ea409c5c0c318bfc4 : SUCCESS in 16m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/390e9ad7ca514020a7fa09c27dc39b9c : SUCCESS in 1h 10m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/024d25575ff447d790718b491392c9ab : SUCCESS in 1h 18m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0493cd0ff7f542d48649858abcd5eec1 : SUCCESS in 16m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/07206b8c8c5c451793ca0fcb09b28ba5 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be5a0844c89a46fe9a33a2e3615b4713 : SUCCESS in 6m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/af7971197bc84a828042f66f48bef206 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/06f8239aa60543d3b5ddc54d73c41865 : SUCCESS in 40m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f6c571ccfccb497ba9cbc0ae8f7d351a : FAILURE in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0c997b2d1cd9414497f68b6d81da91f8 : SUCCESS in 1h 06m 37s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8a0ff0d253644f3a9a1a1bed2875154d : SUCCESS in 1h 42m 47s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d55ae6b6dc94e929004ac64eac523de : SUCCESS in 1h 13m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eef70d9d9525477aac11464643e9ee45 : SUCCESS in 1h 14m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/92b4c034d7d040459e49a3d9dec270ce : SUCCESS in 1h 28m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7dc99dbe6ddc430bb34ce8a8915619a6 : SUCCESS in 2h 13m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/015ec8db1f8442ad81f6200f3bdeb869 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":8},{"id":"ab2a190fb6fca7df4b5a3d2b2ab840690fd2a650","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-31 08:57:23.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/707390/8/upstream-solidfire-check/cinder-solidfire-iSCSI/9eaa820/ : SUCCESS in 57m 27s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"05549dff48b31b53b3fb72d1f24c263a7614aabd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-31 09:37:17.000000000","message":"Patch Set 8:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-90-707390-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"2455243b0337ffec5d36febcf200618b20ba8ee6","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-31 09:52:13.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/90/707390/8/check/synology-dsm-cinder-iscsi/81e652e : UNSTABLE in 59m 36s","accounts_in_message":[],"_revision_number":8},{"id":"d62219f3a2e83e8746bfdb3658b9103e5c0b130b","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-31 10:32:00.000000000","message":"Patch Set 8:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3579 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"ec2ec4d1e8a1aefe812d75f4018ca8a2d941a30e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-03-31 14:01:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/707390/8/check/DellEMC_SC_FC/85ab908/DellEMC_SC_FC/513 : SUCCESS in 1h 10m 44s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/707390/8/check/DellEMC_SC_iSCSI/36a2883/DellEMC_SC_iSCSI/513 : FAILURE in 2h 01m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"9eed80116043a6a5ceca0af58851e355f4079bc0","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-03-31 14:34:38.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/707390-8/Seagate-iSCSI : SUCCESS in 43:02.67\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/707390-8/Seagate-FC : SUCCESS in 38:37.62","accounts_in_message":[],"_revision_number":8},{"id":"a010a1d49eb30074c0af7788746973dd175b8dd4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-31 15:02:32.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/8/check/DellEMC_XtremIO_FC/ecb5657/DellEMC_XtremIO_FC/904 : SUCCESS in 1h 31m 02s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/8/check/DellEMC_XtremIO_iSCSI/03a08f1/DellEMC_XtremIO_iSCSI/824 : FAILURE in 2h 46m 18s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"f8cbf08deedd70e25cd9275751c70c59d9038fa6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-31 16:30:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/707390/8/upstream-check/cinder-cDOT-iSCSI/a9e1e90/ : SUCCESS in 46m 40s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/707390/8/upstream-check/cinder-cDOT-iSCSI-IPv6/453884c/ : SUCCESS in 2h 03m 34s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/707390/8/upstream-check/cinder-cDOT-NFS/3ae2839/ : SUCCESS in 3h 09m 36s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/707390/8/upstream-check/cinder-cDOT-NFS-IPv6/2177af8/ : FAILURE in 4h 00m 16s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/707390/8/upstream-check/cinder-cDOT-FCP/4c89564/ : SUCCESS in 1h 32m 58s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"2b89173ded8287e6994b395b59c6b13819b22942","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-31 16:44:41.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/8/check/tempest-dsvm-full-bionic/958708a/ : SUCCESS in 1h 29m 03s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/8/check/devstack-plugin-ceph-dsvm-bionic/420d704/ : SUCCESS in 1h 53m 03s","accounts_in_message":[],"_revision_number":8},{"id":"769f04277cbcc28b51e4628b21075d05e8715318","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-31 17:31:23.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/90/707390/8/check-not-vote/ext-cinder-zuul/58c79a1 : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":8},{"id":"f202062ddc255589c5532f7d483629633e6a488c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-31 18:33:28.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/8/check/DellEMC_VxFlexOS/200797b/DellEMC_VxFlexOS/521 : SUCCESS in 2h 14m 57s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"ddf619833eab90c8bad1fb3f2c060c046ffc1d51","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-03-31 23:23:02.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/586/index.html : SUCCESS in 1h 07m 09s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/731/index.html : FAILURE in 16s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/774/index.html : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":8},{"id":"e42500dca2fa67b8f56c497c14bd5fd62471c0e8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-04-01 00:01:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/8/20200401063514/NEC-FC : SUCCESS in 2h 17m 14s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/8/20200401063514/NEC-ISCSI : SUCCESS in 2h 25m 55s","accounts_in_message":[],"_revision_number":8},{"id":"09d23a5805f0d137750b46efa4b70c53147f58e2","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-04-01 00:54:30.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/707390-8/Lenovo-iSCSI : SUCCESS in 46:59.90\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/707390-8/Lenovo-FC : SUCCESS in 45:10.50","accounts_in_message":[],"_revision_number":8},{"id":"57f279c6694e82531b69d00fbbc4ba3801ba9360","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-01 01:57:40.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/707390/8/check/DellEMC_VNX_iSCSI/5b3fcba/DellEMC_VNX_iSCSI/463 : FAILURE in 2h 52m 12s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/707390/8/check/DellEMC_VNX_FC/fd34334/DellEMC_VNX_FC/461 : SUCCESS in 1h 00m 14s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"de45f51f8bb78ed5e323a1e7c66c21204fec63fa","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-01 04:54:51.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"566d5e03f55de4e45774717b51b2946c3e443d68","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-01 09:28:18.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"73c2535451f80033f8537e3cb1cc479e4f476d0b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-01 10:43:51.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"9c724f0b7c5c211603f2c24b4f47c244dbf3e743","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-04-01 11:31:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bf8d9a7a2e78cd3ad19d97ce085d559760f7c4dc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-01 11:32:26.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"6c08fdddbb924f2e2e043a59246d8040396826f0","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-01 11:33:43.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/9/iscsi/ : FAILURE in 53s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/9/smb/ : FAILURE in 52s","accounts_in_message":[],"_revision_number":9},{"id":"e7d73d53c2d41b71a86edc3efeff486b3ea40552","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-01 12:11:30.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/90/707390/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8adf235/8adf235.html : FAILURE in 38m 05s","accounts_in_message":[],"_revision_number":9},{"id":"e9641014be07bac8c109abe20eefd3338248f9f6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-01 12:36:26.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/9/check/fujitsu-eternus-dx-iscsi/0ba05e2 : FAILURE in 30m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/9/check/fujitsu-eternus-dx-fc/5e556f1 : FAILURE in 30m 18s","accounts_in_message":[],"_revision_number":9},{"id":"28851501c2337c5bcce8a0eeabb15832f5e188cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 13:55:25.000000000","message":"Patch Set 9: Verified+1\n\n(2 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/076c572e12d140fa92f42da5e2859756 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79594d24ff2b494795f4c163e48e01b6 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f302f7e00d8493c93cc1b4b774bed93 : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15c8114a7b2a4f1188dc719e14822841 : SUCCESS in 13m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52df5e0c78db409f8eba9448c7e8a40e : FAILURE in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abcb88276c548f8b9667de01ccabe1e : SUCCESS in 13m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/09f2b2bdd3c446ceb4b794178d6faf83 : SUCCESS in 1h 00m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c43ee44d0f44c01bab4b8c261a78a12 : SUCCESS in 1h 17m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/542496ca05f249e9ac7436589bb8e5f1 : SUCCESS in 14m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b57b0af84e2e43b99f95a78daaaa38e4 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b54099a7f1d43adb32f216ed9f9bb05 : SUCCESS in 5m 33s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f6e2981eb80f4155b630f9b1d2b9affd : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d3595764d6bf4bd293356c4a816bac6f : SUCCESS in 37m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ca6985361574ca79e9a6d9a91386ce6 : FAILURE in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e3c6db691c3d47209275bfd4bca39d87 : SUCCESS in 1h 00m 28s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/297adcbc236347608c100ff6d4b7b0b8 : SUCCESS in 1h 21m 19s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eee4650762a54e83b26ffc040eb09898 : SUCCESS in 1h 01m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4291666a324e4a97a04d65f89d10383b : SUCCESS in 1h 12m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/876f0935eb174ccd9a394315319ed5bf : SUCCESS in 1h 37m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5d7d60a27f724b35a8e3f202946d5905 : SUCCESS in 2h 11m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f50ab2dc19c7489fbdbb10ebd934c768 : SUCCESS in 53m 33s\n\nWarning:\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Comments left for invalid file stdin","accounts_in_message":[],"_revision_number":9},{"id":"03f9f6980cf10c072a06e5e295eaf9819bfb3dcc","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-01 14:07:53.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/9/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ffb0c27 : FAILURE in 1h 58m 42s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/9/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/bdc7118 : FAILURE in 2h 07m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"7a296c0fe589e312912be7641e1394eb264ab8d6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-04-01 14:57:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/707390/9/upstream-solidfire-check/cinder-solidfire-iSCSI/d0a3a3f/ : SUCCESS in 1h 00m 21s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"7eab9f851b62e6d724dfc5b951d278be84e5a7b7","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-04-01 15:42:43.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/90/707390/9/check/synology-dsm-cinder-iscsi/5ecbf38 : UNSTABLE in 56m 46s","accounts_in_message":[],"_revision_number":9},{"id":"b243dbe452c0091eaf29a45b079cfbc69378d8ef","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-01 16:05:20.000000000","message":"Patch Set 9:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-90-707390-9 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":9},{"id":"0a9528ede80d27a8720bfe484e0df9798b6b6d41","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-04-01 16:31:55.000000000","message":"Patch Set 9:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3599 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"de82672ef9c4e6fff1a65df2b13b5f1e8b02f6f5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-01 17:10:54.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/9/check/tempest-dsvm-full-bionic/e84ac67/ : SUCCESS in 1h 57m 01s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/9/check/devstack-plugin-ceph-dsvm-bionic/e362763/ : SUCCESS in 1h 56m 48s","accounts_in_message":[],"_revision_number":9},{"id":"af06ace6b2c7a47403050fa552632ddae3933551","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-01 21:42:21.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/707390/9/check/DellEMC_Unity_FC/7ebba11/DellEMC_Unity_FC/824 : SUCCESS in 1h 26m 57s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/707390/9/check/DellEMC_Unity_iSCSI/9f63f99/DellEMC_Unity_iSCSI/794 : SUCCESS in 1h 48m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"a1055b8f3c32a22d4830c3327004a79f1f8e86b4","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-01 22:07:25.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/9/check/nimble-iscsi-driver-dsvm-volume/df56c44 : SUCCESS in 1h 40m 41s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/90/707390/9/check/nimble-fc-driver-dsvm-volume/70513d5 : SUCCESS in 1h 18m 54s","accounts_in_message":[],"_revision_number":9},{"id":"03f48cabbc52fc74a852e4e43e1268636857b7df","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-01 22:40:04.000000000","message":"Patch Set 9:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-9/ns5_nfs : SUCCESS in 7h 58m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-9/ns5_iscsi : SUCCESS in 8h 8s","accounts_in_message":[],"_revision_number":9},{"id":"8aca63867ddd1f584b401bc3b06130f309346e43","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-04-02 03:08:30.000000000","message":"Patch Set 9:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/707390-9/Seagate-iSCSI : SUCCESS in 38:29.20\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/707390-9/Seagate-FC : SUCCESS in 35:15.46","accounts_in_message":[],"_revision_number":9},{"id":"dbdb73badfd744268e9eed48f39ff5ece887fdd6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-04-02 05:56:37.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/707390/9/check/DellEMC_SC_FC/86008b9/DellEMC_SC_FC/532 : SUCCESS in 1h 09m 57s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/707390/9/check/DellEMC_SC_iSCSI/b5f4918/DellEMC_SC_iSCSI/531 : SUCCESS in 2h 11m 12s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"2e19fa33a600c134cd84551cc8c5b04f4c92ac09","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-02 10:33:04.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/90/707390/9/check-not-vote/ext-cinder-zuul/2a53664 : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":9},{"id":"d1ca2bab16483f818e8b07f2da8551850b6499d7","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-02 11:14:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/609/index.html : SUCCESS in 1h 12m 39s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/754/index.html : SUCCESS in 1h 34m 15s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/798/index.html : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":9},{"id":"c0880b8900c52ebd3cf5bcc7210ee04fecb096cb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-02 11:32:09.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/9/check/DellEMC_VxFlexOS/5c8237c/DellEMC_VxFlexOS/539 : SUCCESS in 2h 15m 53s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":9},{"id":"5428c8d7a37fc1d2ba866fcf06d1520ee55af473","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-02 15:44:29.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/707390/9/check/DellEMC_XtremIO_FC/f74dd4d/DellEMC_XtremIO_FC/930 : SUCCESS in 1h 35m 08s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/707390/9/check/DellEMC_XtremIO_iSCSI/c6369ff/DellEMC_XtremIO_iSCSI/847 : FAILURE in 4h 26m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"6e23f68b15bc9cb20be9234fd31da7c0cb98d027","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-04-02 16:57:55.000000000","message":"Patch Set 9:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/707390-9/Lenovo-iSCSI : SUCCESS in 47:56.85\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/707390-9/Lenovo-FC : SUCCESS in 49:22.42","accounts_in_message":[],"_revision_number":9},{"id":"5bbd87aba53afaecbcbf767b1d42420bfa2a749e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-02 21:31:20.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/707390/9/check/DellEMC_VNX_iSCSI/e523c4a/DellEMC_VNX_iSCSI/479 : FAILURE in 2h 51m 52s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/707390/9/check/DellEMC_VNX_FC/765d062/DellEMC_VNX_FC/480 : SUCCESS in 54m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"933aa1ff4621149422c3014684e050bc22892301","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-04-02 22:00:02.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/9/20200403043227/NEC-FC : SUCCESS in 2h 21m 55s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/9/20200403043227/NEC-ISCSI : SUCCESS in 2h 27m 33s","accounts_in_message":[],"_revision_number":9},{"id":"2d38a58c1c8009e3124dc6e744d11fd233f83043","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-04-03 15:21:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cd9cc02bd49d3ac91fba20691ec6ad5969e5f685","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-03 16:01:33.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/90/707390/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/48e936c/48e936c.html : FAILURE in 38m 03s","accounts_in_message":[],"_revision_number":10},{"id":"8044bebdb0210089b03336363cbeed7bf5d99207","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-03 16:27:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/707390/10/check/fujitsu-eternus-dx-iscsi/163b801 : FAILURE in 31m 05s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/707390/10/check/fujitsu-eternus-dx-fc/dbef6aa : FAILURE in 30m 49s","accounts_in_message":[],"_revision_number":10},{"id":"6f0724212cdd313ae6bf9d10b63b9a8a740b680e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-03 16:27:59.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/707390/10/iscsi/ : SUCCESS in 1h 05m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/707390/10/smb/ : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":10},{"id":"534fb7325289e6221bdc70e104c18bcb42b7838e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-03 16:46:13.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/90/707390/10/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/32b2949 : UNSTABLE in 1h 02m 06s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/90/707390/10/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/92572e8 : UNSTABLE in 1h 22m 32s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"5a1096cba90e16cf8037ad190b8949d7803adbee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-03 17:09:36.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/10/check/tempest-dsvm-full-bionic/d963c8a/ : SUCCESS in 1h 40m 38s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/707390/10/check/devstack-plugin-ceph-dsvm-bionic/35e7a85/ : SUCCESS in 1h 34m 38s","accounts_in_message":[],"_revision_number":10},{"id":"0d585ae111ed967bce9d6ac96b5604310f8104b6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-03 17:38:48.000000000","message":"Patch Set 10:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-707390-10/ns5_nfs : SUCCESS in 2h 3m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-707390-10/ns5_iscsi : SUCCESS in 2h 16m 56s","accounts_in_message":[],"_revision_number":10},{"id":"a22e92e2490a9570b58e60c4e5c4f08a33ffa9a0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-03 18:37:31.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/707390/10/check/DellEMC_Unity_FC/4b4f708/DellEMC_Unity_FC/847 : SUCCESS in 1h 23m 20s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/707390/10/check/DellEMC_Unity_iSCSI/d103db6/DellEMC_Unity_iSCSI/814 : SUCCESS in 1h 38m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"22c532fa15d1a19a96bdbb20dbd600d7c778ab4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 19:53:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f87fa020b884b8e9c4a6d6fda19c2c6 : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/481f5c8c96314a38a4b8f37711d9eaa0 : SUCCESS in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c03c9e1030954459b162a4b0dd6aaa45 : SUCCESS in 14m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/114dcfd8ba524137bcb49dd2af26ffa6 : SUCCESS in 15m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abaa7fa0f2e14d3196352b8599f44fe3 : FAILURE in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/821e2129b0cc41e7a0f2cbfd3d4fd917 : SUCCESS in 16m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2a19682c50074983a8a44c32046b8208 : SUCCESS in 57m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31d65a6fc72a4c0e9145c1eb892551f8 : SUCCESS in 1h 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8571c7decc81491aa10bb8c443990c76 : SUCCESS in 21m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01a3de3b066a41bda696ec38d8688052 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dbf879536854c78b6d5eaf898e930bc : SUCCESS in 7m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/53f9a1e4ca3941638ca66065acc1ca91 : SUCCESS in 9m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/626406082d0f4eafa0ef0edbdbe9e3a6 : SUCCESS in 45m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/256b0c9ba5a04f59937f4698e0e7f7d9 : FAILURE in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e370bc7a184a4bd79b1cf6f1fa75d125 : SUCCESS in 1h 03m 36s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/98363a117b9549eabdbcda84ed37d2c5 : SUCCESS in 1h 34m 51s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f9fa982d69f84f7b9dcd0f764235c73f : SUCCESS in 1h 15m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8547ae4a77434270b6bb4f84aa9a5ceb : SUCCESS in 1h 25m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8ad7e6d9fb95469b976ce71e7d03909a : SUCCESS in 1h 20m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50926e290ca14986b460aa72ecf3c9c9 : SUCCESS in 2h 22m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49ba76a22d064a1cac2b00493f306ccb : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":10},{"id":"f9a3f67bd0f165156153c9635d45e8fbc35b4217","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-03 21:40:08.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"a4d146a6c12c1fc43d34baba7336bfea8c8a1102","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-04-03 22:30:49.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/90/707390/10/check/synology-dsm-cinder-iscsi/f843d18 : UNSTABLE in 57m 53s","accounts_in_message":[],"_revision_number":10},{"id":"3f415377a883e060fbc99bbdb9f859193ef192d3","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-04-03 22:48:58.000000000","message":"Patch Set 10:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3625 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"033d14bd5034836e915bf575550680acaa65b46a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-04 00:01:49.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"4f02fd49ad03451f44a1bb006f3d6751839f42b1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-04 00:48:07.000000000","message":"Patch Set 10:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-90-707390-10 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":10},{"id":"80eac834b45ff61cf9913e8b18e43e82fa3dfbad","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-04-04 01:00:16.000000000","message":"Patch Set 10:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/707390-10/Seagate-iSCSI : SUCCESS in 38:44.58\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/707390-10/Seagate-FC : SUCCESS in 34:47.32","accounts_in_message":[],"_revision_number":10},{"id":"46ba4d73833018bb2774219839350b554d4891a5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-04 05:16:38.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/707390/10/check/DellEMC_Unity_FC/81c8a66/DellEMC_Unity_FC/853 : SUCCESS in 1h 17m 13s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/707390/10/check/DellEMC_Unity_iSCSI/8510b9b/DellEMC_Unity_iSCSI/820 : SUCCESS in 1h 45m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"41139b13ed33ab596ef787727430474560f5058e","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-04 10:00:31.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/636/index.html : SUCCESS in 1h 16m 56s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/781/index.html : SUCCESS in 1h 44m 52s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/825/index.html : SUCCESS in 1h 29m 40s","accounts_in_message":[],"_revision_number":10},{"id":"92928d8d40b39faf3c8dac1af56fa40c3ddd12dc","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-04-04 16:35:14.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/707390/10/check/DellEMC_SC_FC/5bccdc8/DellEMC_SC_FC/556 : SUCCESS in 1h 06m 37s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/707390/10/check/DellEMC_SC_iSCSI/4430e7f/DellEMC_SC_iSCSI/555 : SUCCESS in 2h 08m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"6fdd425b447cda2c8a67d42cb3f91e758b9f90c4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-04 20:49:45.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/90/707390/10/check/DellEMC_VxFlexOS/bc1e0c8/DellEMC_VxFlexOS/563 : SUCCESS in 2h 13m 58s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":10},{"id":"a6f9bf829f1bd56d2f2070c9d51b934f0e3704d6","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-04-05 06:06:18.000000000","message":"Patch Set 10:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/707390-10/Lenovo-iSCSI : SUCCESS in 50:12.08\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/707390-10/Lenovo-FC : SUCCESS in 42:54.88","accounts_in_message":[],"_revision_number":10},{"id":"f3e30b8a2d94e18c59302f06466c0d3182a9f388","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-04-05 17:02:07.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/10/20200405232450/NEC-ISCSI : SUCCESS in 2h 31m 17s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/707390/10/20200405232450/NEC-FC : SUCCESS in 2h 37m 15s","accounts_in_message":[],"_revision_number":10},{"id":"4cf92ef83a4eeb7cfe54a38d6121ba5ba84217e9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-05 23:25:34.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/707390/10/check/DellEMC_VNX_iSCSI/6264ee5/DellEMC_VNX_iSCSI/503 : FAILURE in 2h 50m 33s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/707390/10/check/DellEMC_VNX_FC/4721441/DellEMC_VNX_FC/504 : FAILURE in 1h 01m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"58aa9605498abb3c1970f781ff76b70600b23e9c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-06 08:30:42.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"6dc38bc29589134b7f3fc18272425140f5be3554","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-06 12:34:20.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"4cff664fb9430dd4e910b1bc65b5a3a4bbeaff98","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-06 21:24:12.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/707390/10/check/DellEMC_PowerMAX_FC/cd0ed78/DellEMC_PowerMAX_FC/590 : SUCCESS in 6h 11m 08s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/707390/10/check/DellEMC_PowerMAX_iSCSI/ae77752/DellEMC_PowerMAX_iSCSI/498 : ABORTED in 2h 42m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":10},{"id":"9b6e317d6aee6392acfc8980d69633ac3530b994","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-04-07 14:40:57.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"5492bc8be3429def34f684ab19b41b12ab61a3ad","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-04-07 14:44:40.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"af9dc8951a95efa8c93efd243413a2d77809b693","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-08 06:09:03.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/90/707390/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"61f129524aba04d4b32c318e79fbd2e1963aba57","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 19:18:17.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"421939da6c9cb25a26fd7590ab9d3595af1d948d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-04-08 19:21:16.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)\n\nPassing 3rd Party CI and well documented.  Thank you.\n\nThere are a couple of nits inline but nothing to hold up the patch.  Can be covered by follow-up.","accounts_in_message":[],"_revision_number":10},{"id":"4d39def4e39010fabb39e047eac20d1efb995d6e","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-04-08 19:33:45.000000000","message":"Patch Set 10:\n\n\u003e (2 comments)\n \u003e \n \u003e Passing 3rd Party CI and well documented.  Thank you.\n \u003e \n \u003e There are a couple of nits inline but nothing to hold up the patch.\n \u003e  Can be covered by follow-up.\n\nHi, Jay.\nThanks for review.\nYeah, seems that debug is more suitable. Will change it in later patches.","accounts_in_message":[],"_revision_number":10},{"id":"1aa9ef99f5292f59aecaa57cbfab62abb84033d6","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-04-08 19:42:39.000000000","message":"Patch Set 10: Workflow+1\n\nSince Sean has looked at this I think it is ok to merge.","accounts_in_message":[],"_revision_number":10},{"id":"36a394a2fe6309c5f4cd5b74619631c511484a18","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-04-08 19:43:11.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"f77bfab9b048aaafdb457270606f82fc70ea0ee9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-04-08 19:44:36.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"80076f0e923584e2fdb76be97403b48d5d977261","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:50:58.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a2f4a461414a5228d6ba41f6765c1676175e9d36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:01:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"931f64e627356e57846c9b8043c83fb4e5cc204e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:01:24.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed178eb46baa4dfdb530f23509e2d774 : SUCCESS in 15m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0a3ce379bb4d5ebbf5d372e64a3e88 : SUCCESS in 9m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60299c51c33a46cf9cb1f15f8da998db : SUCCESS in 12m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbf0fbf869dc4f2683db7f201cd47df0 : SUCCESS in 16m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa33cb676fc49a5887d3eeb89c5b64f : SUCCESS in 17m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8cff687fd4974d3f942725ad6f24b6b7 : SUCCESS in 1h 09m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d411805495ed414d939358b61d8338c8 : SUCCESS in 1h 29m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26b8cd8f3b244e8b90391d498f0c5a19 : SUCCESS in 19m 22s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de9eeed176d44abe9c90436141779342 : SUCCESS in 1h 29m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5f5196c329c44233b9ecb28405035206 : SUCCESS in 1h 01m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55a1897d2bdb461baf48a2bf46c29669 : SUCCESS in 1h 01m 35s\n\nWarning:\n  Comments left for invalid file cinder/scheduler/host_manager.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file cinder/volume/drivers/stx/client.py","accounts_in_message":[],"_revision_number":10},{"id":"72a5aa005cd3a97f2bbafa8fa9c5e83eb04621b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:08:38.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d54d9ad5b744c5bf639a360be88199 : SUCCESS in 2m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f919904da38f4e51ade93f063dc5bb50 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":10},{"id":"7c34211821e33deeeb22587f5355365ce6e78f85","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-13 16:14:21.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"b1e64247cded13079396aadcd79b94feafb28112","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-14 04:30:00.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"a2ebca3167c02f07b345ed36db0305d088ff2589","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-14 15:51:10.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/707390/10 : FAILURE","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"4e320543cf6a115fdaa807ee9df3a8adbbd9e920","revisions":{"13246959c8207a895f762448bf483d091c73383e":{"kind":"REWORK","_number":1,"created":"2020-02-12 13:44:28.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/1"}}},"commit":{"parents":[{"commit":"72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:43.000000000","tz":180},"subject":"Add replication v2.1 support in VxFlex OS driver","message":"Add replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13246959c8207a895f762448bf483d091c73383e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13246959c8207a895f762448bf483d091c73383e"}]},"branch":"refs/heads/master"},"334a9327f91f74b6a98cb00230f45490f38826c2":{"kind":"REWORK","_number":2,"created":"2020-02-13 11:21:11.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/2"}}},"commit":{"parents":[{"commit":"72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-13 11:20:41.000000000","tz":180},"subject":"Add replication v2.1 support in VxFlex OS driver","message":"Add replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/334a9327f91f74b6a98cb00230f45490f38826c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/334a9327f91f74b6a98cb00230f45490f38826c2"}]},"branch":"refs/heads/master"},"eeadb174980df63ab96f05b6c423842124ff510b":{"kind":"REWORK","_number":3,"created":"2020-02-13 11:43:48.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/3"}}},"commit":{"parents":[{"commit":"72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-13 11:41:05.000000000","tz":180},"subject":"Add replication v2.1 support in VxFlex OS driver","message":"Add replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eeadb174980df63ab96f05b6c423842124ff510b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eeadb174980df63ab96f05b6c423842124ff510b"}]},"branch":"refs/heads/master"},"109866f179c3bed989cd1cecce91c5d780dfe66f":{"kind":"REWORK","_number":4,"created":"2020-02-19 10:16:36.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/4"}}},"commit":{"parents":[{"commit":"72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-19 09:15:25.000000000","tz":180},"subject":"Add OpenStack volume replication v2.1 support in VxFlex OS driver","message":"Add OpenStack volume replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS (v3.5.0) release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled according to OpenStack volume replication specification.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/109866f179c3bed989cd1cecce91c5d780dfe66f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/109866f179c3bed989cd1cecce91c5d780dfe66f"}]},"branch":"refs/heads/master"},"00122a9cbb65e037f783505dc11acfaec975743c":{"kind":"REWORK","_number":5,"created":"2020-02-25 09:24:43.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/5"}}},"commit":{"parents":[{"commit":"72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-25 09:12:22.000000000","tz":180},"subject":"Add OpenStack volume replication v2.1 support in VxFlex OS driver","message":"Add OpenStack volume replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS (v3.5.0)\nrelease.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled according to OpenStack volume replication\nspecification.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/00122a9cbb65e037f783505dc11acfaec975743c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/00122a9cbb65e037f783505dc11acfaec975743c"}]},"branch":"refs/heads/master"},"bf1088ef32335c2f7d2a9f5a9e76cd527119074d":{"kind":"REWORK","_number":6,"created":"2020-02-28 10:17:58.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/6"}}},"commit":{"parents":[{"commit":"72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-28 10:17:17.000000000","tz":180},"subject":"Add OpenStack volume replication v2.1 support in VxFlex OS driver","message":"Add OpenStack volume replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS (v3.5.0) release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled according to OpenStack volume replication specification.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf1088ef32335c2f7d2a9f5a9e76cd527119074d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf1088ef32335c2f7d2a9f5a9e76cd527119074d"}]},"branch":"refs/heads/master"},"2ebcb827a58d8174fd02d66baab5ebbf4412fde7":{"kind":"REWORK","_number":7,"created":"2020-03-04 08:45:23.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/7"}}},"commit":{"parents":[{"commit":"72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72e59b7d4e58d3d4667f0e0122a3c58cdb7c15fd"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-03-04 08:44:57.000000000","tz":180},"subject":"Add OpenStack volume replication v2.1 support in VxFlex OS driver","message":"Add OpenStack volume replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS (v3.5.0) release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled according to OpenStack volume replication specification.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ebcb827a58d8174fd02d66baab5ebbf4412fde7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ebcb827a58d8174fd02d66baab5ebbf4412fde7"}]},"branch":"refs/heads/master"},"3d1f44d0de190057a7ef56aeb5f246108f24ee59":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-03-31 05:42:41.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/8"}}},"commit":{"parents":[{"commit":"010568f1654495b0e8227f0b4e50965739c9add2","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/010568f1654495b0e8227f0b4e50965739c9add2"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-03-31 05:21:29.000000000","tz":180},"subject":"Add OpenStack volume replication v2.1 support in VxFlex OS driver","message":"Add OpenStack volume replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS (v3.5.0) release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled according to OpenStack volume replication specification.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d1f44d0de190057a7ef56aeb5f246108f24ee59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d1f44d0de190057a7ef56aeb5f246108f24ee59"}]},"branch":"refs/heads/master"},"b872c73d69d284ede156c7c60f6729ee9dd0013c":{"kind":"REWORK","_number":9,"created":"2020-04-01 11:31:09.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/9"}}},"commit":{"parents":[{"commit":"b1e49b801bb153e4fdb68cc8e342ac69a8ab64b3","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1e49b801bb153e4fdb68cc8e342ac69a8ab64b3"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-04-01 11:24:12.000000000","tz":180},"subject":"Add OpenStack volume replication v2.1 support in VxFlex OS driver","message":"Add OpenStack volume replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS (v3.5.0) release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled according to OpenStack volume replication specification.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b872c73d69d284ede156c7c60f6729ee9dd0013c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b872c73d69d284ede156c7c60f6729ee9dd0013c"}]},"branch":"refs/heads/master"},"4e320543cf6a115fdaa807ee9df3a8adbbd9e920":{"kind":"REWORK","_number":10,"created":"2020-04-03 15:21:31.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/707390/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/707390/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/707390/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/707390/10"}}},"commit":{"parents":[{"commit":"f75b2865fe2aee3f6aefea8f08ce92e1d590ac98","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f75b2865fe2aee3f6aefea8f08ce92e1d590ac98"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-02-12 13:43:08.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-04-03 15:20:39.000000000","tz":180},"subject":"Add OpenStack volume replication v2.1 support in VxFlex OS driver","message":"Add OpenStack volume replication v2.1 support in VxFlex OS driver\n\nVolume replication support will be added in next VxFlex OS (v3.5.0) release.\nCinder driver for VxFlex OS supports volumes/snapshots with\nreplication enabled according to OpenStack volume replication specification.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I3f2ec1ddf1867261253190953f7a65fff22c7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e320543cf6a115fdaa807ee9df3a8adbbd9e920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e320543cf6a115fdaa807ee9df3a8adbbd9e920"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
