)]}'
{"id":"openstack%2Fcinder~713649","triplet_id":"openstack%2Fcinder~master~I6f160cc44350f0d378fdddb7943e8cdcc15be1b8","project":"openstack/cinder","branch":"master","topic":"a-a-support","hashtags":[],"change_id":"I6f160cc44350f0d378fdddb7943e8cdcc15be1b8","subject":"Set cluster name for volume groups","status":"MERGED","created":"2020-03-18 13:11:23.000000000","updated":"2020-08-07 06:26:43.000000000","submitted":"2020-08-07 06:26:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713649-1596781603009-07a05074","meta_rev_id":"20550126a84f82a575d245318ff39526d4ec679f","_number":713649,"virtual_id_number":713649,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-22 12:29:03.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-07-22 15:48:27.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2020-07-23 13:54:53.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 01:16:11.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 10:17:01.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-04 13:45:32.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 15:27:56.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-07 06:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 03:53:32.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-07-22 15:25:10.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2020-07-23 15:05:02.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-07-22 10:18:26.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 20:30:30.000000000","_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 17:27:02.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 20:39:43.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"date":"2020-07-23 03:03:20.000000000","_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-07-22 08:58:25.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 15:45:05.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"date":"2020-07-22 09:05:11.000000000","_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 20:00:14.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 02:55:59.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"value":0,"date":"2020-07-23 03:40:58.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2020-07-22 17:15:35.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 19:32:01.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 12:08:54.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"date":"2020-07-23 11:13:51.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-07 02:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":1,"date":"2020-07-23 01:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-22 07:47:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-07 02:16:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 21:23:10.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-03-25 14:48:27.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-03-26 11:38:29.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"REVIEWER"},{"updated":"2020-03-27 19:02:43.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-27 20:31:43.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-29 15:50:24.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2020-04-29 16:42:12.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-30 12:41:19.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"REVIEWER"},{"updated":"2020-04-30 14:37:46.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-30 15:18:47.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-04-30 21:00:39.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-01 03:12:58.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-05-02 17:37:03.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2020-05-02 22:15:29.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 08:58:25.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 09:05:11.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 10:18:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 12:29:03.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 15:25:10.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 15:27:56.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 15:45:05.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-07-22 15:48:27.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-07-22 17:15:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 17:27:02.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 19:32:01.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 20:00:14.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 20:30:30.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 20:39:43.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-07-23 01:16:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 01:46:42.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-07-23 02:55:59.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-07-23 03:03:20.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-07-23 03:40:58.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-07-23 03:53:32.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-07-23 10:17:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 11:13:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 12:08:54.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 13:54:53.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 15:05:02.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-08-07 02:16:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-07 06:26:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76788d937ea6d37719ee4dd90a513b1771226488","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-18 13:11:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07d52f18415c80b50b608e8e002a29356d9ae021","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-18 13:44:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/1/check/fujitsu-eternus-dx-iscsi/aa86a96 : FAILURE in 32m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/1/check/fujitsu-eternus-dx-fc/b2c2201 : FAILURE in 31m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7af886c2f9c89facb741fdf902405cae43d71506","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-18 13:44:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/1/Cinder-tgtadm : ABORTED\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/1/Cinder-lioadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"0ff2dedada2f4c0d0806d7c9dda9a43f259a0da7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-18 14:11:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/1/iscsi/ : SUCCESS in 59m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/1/smb/ : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7d509751571dea09890e1a3992b49acc0921f55d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-18 14:28:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9fd5c3f : UNSTABLE in 1h 00m 23s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/2994fd8 : UNSTABLE in 1h 16m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"b15c9aca8301d289dfe710eadc9872a56067327f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-18 14:51:17.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-1/ns5_nfs : SUCCESS in 1h 36m 58s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-1/ns5_iscsi : SUCCESS in 1h 39m 27s","accounts_in_message":[],"_revision_number":1},{"id":"54299287b34680a3809333aecd61262614ad60d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-18 14:51:55.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/1/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/1/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"fff44281ab4a903a26edab0b93d0223b302487c6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-18 15:17:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/1/check/tempest-dsvm-full-bionic/2901e5b/ : SUCCESS in 1h 47m 50s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/1/check/devstack-plugin-ceph-dsvm-bionic/fe546b0/ : SUCCESS in 1h 44m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f81dbb41ffd42f69d1b63041f5be19b70537b2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 15:56:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca61855e80e64cc080ce30ee9cc3b8b7 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006931dfe7c14c5282e9d9f047e70e35 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d2531950f3d40b78a7403e1eb825943 : SUCCESS in 10m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8065cd84ccb64914b8b7520e2b1965fe : SUCCESS in 13m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c402793d96da4f9baa62076b23320ad9 : FAILURE in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18aedca3e87542989df2afcdaa97d660 : SUCCESS in 15m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3c90279279874ab8bec262e0dfdfbbed : SUCCESS in 1h 05m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca45a678aa034fa7ba212e53e508d5d6 : SUCCESS in 1h 30m 17s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6b89afed5c9b4695b35227ec3a2e12f4 : SUCCESS in 1h 14m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/29b8e6ede38d4326b2056d4e68f5c3df : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53a11376a082495e986549fa8e7ef6d6 : SUCCESS in 6m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a84e35aaba864238a9569918f8f4e96e : SUCCESS in 8m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a5ef95f608c43e7bc095b735241b53a : SUCCESS in 41m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f88fa8d4df9746309c1593155f5447fb : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3813b80e426049d5902e645abfc23050 : SUCCESS in 59m 54s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/136aa3f649df4e7d8950adf6cb4a168c : SUCCESS in 1h 14m 29s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bddde9aedb4347c0ade0f39b35232153 : SUCCESS in 1h 30m 09s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/b68a0e8160bb4c5bb869c4b4af089f20 : SUCCESS in 1h 24m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a80138f9d13b4a4da014e132110876e1 : SUCCESS in 2h 26m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41f44f2e70bf4e13aa4b700d60e5178e : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":1},{"id":"85f88893b1b43aa3e1d5426d62d3d85b7b59599f","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-18 16:07:46.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/49/713649/1/check/synology-dsm-cinder-iscsi/5684f7e : FAILURE in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4a47c1e16ea8dff06f26825b829f0de19be3b5a9","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-18 16:08:32.000000000","message":"Patch Set 1:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3485 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bc728df14ffe747f9d241ca138f03f84e0191d27","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-18 16:30:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/1/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/1/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9347aba001547e05a90f0996450f59478eb0964a","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-03-18 16:40:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/467/index.html : SUCCESS in 1h 13m 06s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/613/index.html : SUCCESS in 1h 37m 25s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/653/index.html : SUCCESS in 1h 23m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fd2643c4e47a39412d5e89e898a07035f4fa3ece","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-18 16:55:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/49/713649/1/upstream-solidfire-check/cinder-solidfire-iSCSI/3651a57/ : SUCCESS in 56m 19s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f96a8f62bc52ac58bf084d624c8b7bdec13ae56","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 16:58:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/1/20200318221125/NEC-FC : SUCCESS in 3h 33m 23s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/1/20200318221125/NEC-ISCSI : SUCCESS in 3h 47m 06s","accounts_in_message":[],"_revision_number":1},{"id":"21e3fc080d7911f1e0bb5c5368c111d44795cd8f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-03-18 17:13:08.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"93ffcca4df09f75c86d6703d8d59ac9b5b4edee9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-18 17:15:51.000000000","message":"Patch Set 1:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-49-713649-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"75cb639210e1040b662b87e22ba977668670422c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 17:16:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/49/713649/1/check/DellEMC_VNX_iSCSI/d8751e7/DellEMC_VNX_iSCSI/409 : SUCCESS in 1h 19m 06s\n- DellEMC_VNX_FC http://publiclogs.emc.com/49/713649/1/check/DellEMC_VNX_FC/7e70572/DellEMC_VNX_FC/410 : FAILURE in 54m 45s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"451bbc589de4086ff19bd7ce485f5b3a49c94a17","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-18 17:26:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/49/713649/1/upstream-check/cinder-cDOT-iSCSI/bafe582/ : SUCCESS in 45m 41s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/49/713649/1/upstream-check/cinder-cDOT-NFS/ba37a5c/ : SUCCESS in 45m 15s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/49/713649/1/upstream-check/cinder-cDOT-FCP/08bb139/ : FAILURE in 16m 42s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e04962b3448ba235cbaeecd699db16ea0211c962","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 18:04:37.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/1/check/DellEMC_XtremIO_FC/961df37/DellEMC_XtremIO_FC/787 : SUCCESS in 1h 29m 01s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/1/check/DellEMC_XtremIO_iSCSI/dd1ce97/DellEMC_XtremIO_iSCSI/736 : SUCCESS in 1h 33m 42s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"bf7e7c8047d5bfb736bb54be79a96d8c0aa482d5","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-03-18 18:33:10.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/713649-1/Seagate-iSCSI : SUCCESS in 43:01.89\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/713649-1/Seagate-FC : SUCCESS in 41:49.92","accounts_in_message":[],"_revision_number":1},{"id":"c5064e5532aeafc7773d8c5d6c468c0b148e66ec","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-18 18:41:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"24097113654d3a61cd871bfd12ec822e4b970fec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-18 18:42:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/1/Cinder-tgtadm : ABORTED\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/1/Cinder-lioadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"115b1dbed0bc3528e9026bd468147c239ca05879","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-03-18 19:21:10.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/713649-1/Lenovo-iSCSI : SUCCESS in 47:11.03\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/713649-1/Lenovo-FC : SUCCESS in 44:22.61","accounts_in_message":[],"_revision_number":1},{"id":"96a8b1c1e03725e3708951240b908896128813ed","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-18 19:53:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/2/iscsi/ : SUCCESS in 57m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/2/smb/ : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":2},{"id":"fec1549bd6719f5c8581b22175e91c7fd38a9c30","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-18 20:08:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/2/check/fujitsu-eternus-dx-iscsi/a7da300 : FAILURE in 30m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/2/check/fujitsu-eternus-dx-fc/fd97d9e : FAILURE in 30m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ea6e3c89ffbe43c75a7515366bbc679617c474e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-18 20:09:07.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/2/Cinder-tgtadm : ABORTED\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/2/Cinder-lioadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"125c95d7df07cc230b732d3ea0ff7a63fc767253","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-18 20:19:49.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"50080bf53313690feb662311e9a13449d913a2ae","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-18 20:45:51.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-2/ns5_nfs : SUCCESS in 1h 52m 12s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-2/ns5_iscsi : SUCCESS in 2h 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8f521c47d8b564c77379bcf2691507ba07604db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 21:02:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd84c146aac24783a809886175b7ac44 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c61e3b7ee0a43808ef77c2cc9910e8f : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adf55ca4753c4c9c8647ab6c45feaa5b : SUCCESS in 8m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0b041c45d034140a5546ee1d1eb64df : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/191fd0b32c0e478781ae8f6a5cf81b1c : FAILURE in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406d06c0ec764aabb1d4178e678f8f4c : SUCCESS in 13m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/69892f91c4b8473fbc538251725dbebf : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62ac0e0d0fb74f18b96dbb1f07710cad : SUCCESS in 1h 24m 43s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bd54c2852915431ca8608518d5a3ae10 : SUCCESS in 1h 32m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f052ddb007840f3b684066c71493389 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f57fde5d2bc4e44ab36da24f30b06af : SUCCESS in 5m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fb7575c296f042babf58326bb81112e9 : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/64e6a5b407624b0c8b3d7dae690aa7f3 : SUCCESS in 53m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b4426b6cd284dac818d190f847407e4 : SUCCESS in 4m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3f298e80f7d4555ad8c63978c9cef8f : SUCCESS in 1h 13m 55s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5910d30d011742febf274979f54d37ea : SUCCESS in 1h 10m 12s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fe6f8df91097436a8a0ed22fa1d51933 : SUCCESS in 1h 33m 06s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/e0fb99ba21b34f939c51680acccff870 : SUCCESS in 1h 33m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab39eb86f5164f769bbb7146794da7ec : SUCCESS in 2h 15m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a47223a8e7bf432f81587c8248440947 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b6f0712850849feb80a285a31f6aa2f710466e2c","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-18 21:13:14.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/49/713649/2/check/synology-dsm-cinder-iscsi/fcffdb5 : FAILURE in 10m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e950f71e9dd65cdac431b26c820afcc09c9b6859","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-18 21:20:20.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/2/check/tempest-dsvm-full-bionic/a5ce089/ : SUCCESS in 1h 45m 29s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/2/check/devstack-plugin-ceph-dsvm-bionic/a575072/ : SUCCESS in 1h 43m 08s","accounts_in_message":[],"_revision_number":2},{"id":"5c1fca318a8e4d62cc6aeebc28f033519def6949","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-18 21:23:00.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"01230f658ee2dac1271d8ad8cc6556d88b439dae","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-18 21:37:12.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/0d31c62 : FAILURE in 1h 57m 49s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/0c2a21a : FAILURE in 1h 57m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"a75b8ce4c173ceab7487775540dec45dddc22d03","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-18 21:59:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/49/713649/2/upstream-solidfire-check/cinder-solidfire-iSCSI/72d26ee/ : SUCCESS in 55m 01s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"5332b6268fe5331d3882eaae38cfddf1a4b91ad6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-18 22:30:07.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/2/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/2/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"aee07498fec06303c3061d1b3082ed240c2229d1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-18 23:15:09.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-49-713649-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"9d2baf1c470e2d07f77a8ad9e493a9535cd8e1f2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-18 23:37:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/49/713649/2/upstream-check/cinder-cDOT-iSCSI/dcf534e/ : SUCCESS in 45m 15s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/49/713649/2/upstream-check/cinder-cDOT-NFS/d30db93/ : SUCCESS in 48m 29s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/49/713649/2/upstream-check/cinder-cDOT-FCP/5783335/ : FAILURE in 26m 39s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"0c22509e39c685ac75a45e5488ec34d67da9ebac","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-19 00:19:58.000000000","message":"Patch Set 2:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3490 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f6c72de96eb9813978b7aac66e7b01b33dedda6e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-19 01:12:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/49/713649/2/check/DellEMC_VNX_iSCSI/8d2dfc5/DellEMC_VNX_iSCSI/414 : FAILURE in 1h 30m 13s\n- DellEMC_VNX_FC http://publiclogs.emc.com/49/713649/2/check/DellEMC_VNX_FC/4ac320c/DellEMC_VNX_FC/415 : SUCCESS in 55m 39s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"2d914d4fcbc993470ff5837b645c3cd054013a74","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-03-19 01:15:13.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"6e67170177f8b9316efb90e8bbea6c46ca2b1194","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-03-19 01:30:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/472/index.html : SUCCESS in 1h 10m 52s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/618/index.html : SUCCESS in 1h 35m 57s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/658/index.html : SUCCESS in 1h 24m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ce84a37b9a40729eb70557957cd8c3dee0ce5c2b","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-03-19 03:34:33.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/713649-2/Seagate-iSCSI : SUCCESS in 39:39.52\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/713649-2/Seagate-FC : SUCCESS in 36:35.29","accounts_in_message":[],"_revision_number":2},{"id":"4aa9aa4609f5ed22cc084f41925460542f2c996b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-19 03:39:51.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/49/713649/2/check/DellEMC_VxFlexOS/362187e/DellEMC_VxFlexOS/452 : SUCCESS in 2h 11m 30s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"ef4c32441d076bcc9e0494af4017ad3480286b63","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-19 03:42:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/2/check/DellEMC_XtremIO_FC/6199175/DellEMC_XtremIO_FC/792 : SUCCESS in 1h 32m 07s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/2/check/DellEMC_XtremIO_iSCSI/f1e20cb/DellEMC_XtremIO_iSCSI/741 : FAILURE in 2h 18m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"2e6a6b50d838b38dceac99ee76c628e3b8c289a5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-19 04:05:26.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/49/713649/2/check/DellEMC_Unity_FC/8593ef7/DellEMC_Unity_FC/698 : SUCCESS in 1h 27m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/49/713649/2/check/DellEMC_Unity_iSCSI/8f7487b/DellEMC_Unity_iSCSI/679 : SUCCESS in 1h 46m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7db0de3211642242bbdb3a22ce012b42c934075c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-19 04:32:47.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/2/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/2/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"df9652b85e771c32e5968a8b644be3a3f7d9ef91","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-19 05:47:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/49/713649/2/check/dsvm-tempest-infinibox-fc/9351861 : SUCCESS in 1h 07m 13s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/49/713649/2/check/dsvm-tempest-infinibox-iscsi/14f0022 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2c582c4c75a58550dc712a912b27c85adefad086","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-03-19 05:55:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/2/20200319124516/NEC-FC : SUCCESS in 2h 09m 41s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/2/20200319124516/NEC-ISCSI : SUCCESS in 2h 10m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bb1c490887baa1afd938188455834fbe26d8107a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-19 05:57:40.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"91a1e56dd8eca55b493b8aad6ece5f22a3886ec3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-03-19 06:05:03.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/713649-2/Lenovo-iSCSI : SUCCESS in 52:59.42\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/713649-2/Lenovo-FC : SUCCESS in 46:20.83","accounts_in_message":[],"_revision_number":2},{"id":"4b631f6339b43e6e03e695c0b8f11d27e5310563","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-03-19 09:45:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/49/713649/2/check/DellEMC_SC_FC/9361a2c/DellEMC_SC_FC/446 : SUCCESS in 1h 26m 40s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/49/713649/2/check/DellEMC_SC_iSCSI/bdcd377/DellEMC_SC_iSCSI/447 : FAILURE in 2h 49m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"48546ca93705e01eac80f9a4a54c8445af8c467f","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-03-19 09:48:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nThis is not the right solution, since the scheduler is doing the right thing as mentioned in the inline comment.\n\nThe bug is most likely in `cinder/group/api.py`, where method\u0027s `_create_group_from_group_snapshot` [1] and `_create_group_from_source_group` [2] set the `host` of the volume but not the `cluster_name`, making those volumes not belong to the cluster. \n So it makes sense that the scheduler cannot find the right backend.\n\n[1]: https://opendev.org/openstack/cinder/src/branch/master/cinder/group/api.py#L342\n[2]: https://opendev.org/openstack/cinder/src/branch/master/cinder/group/api.py#L425","accounts_in_message":[],"_revision_number":2},{"id":"4a2c4498f1fc313bd37950296ab0e3ce319c52a6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-19 09:52:30.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"43429a62868a43801382a0942daac38cb1245c68","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-03-19 10:27:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/49/713649/2/check/cinder-storpool-tempest/69eec75/ : SUCCESS in 3h 00m 52s","accounts_in_message":[],"_revision_number":2},{"id":"763e57ccedb708485462619742cd72c32639c77e","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-03-19 11:00:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/49/713649/2/check/kaminario-dsvm-tempest-full-iscsi/536e997 : SUCCESS in 2h 14m 22s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/49/713649/2/check/kaminario-dsvm-tempest-full-FC/72e1777 : SUCCESS in 1h 59m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d6043321a14e03e2357aa391dbb8dcda49f671ee","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-19 11:32:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"be6e50455767357af5e6ed50872c7e89db183f65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-19 11:34:06.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/2/Cinder-lioadm : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a88ee2b4f3fff65ba2206bbac5e8ba8d26c34be4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-19 12:05:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/3/check/fujitsu-eternus-dx-iscsi/efb7c6e : FAILURE in 31m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/3/check/fujitsu-eternus-dx-fc/66720b8 : FAILURE in 32m 03s","accounts_in_message":[],"_revision_number":3},{"id":"a9d8dad9327a1093bc05f8b7c6f99424e2a40d39","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-19 12:39:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/3/iscsi/ : SUCCESS in 1h 05m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/3/smb/ : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":3},{"id":"451da461417ab7f5a36c62d0b2a432b367560e71","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-19 12:52:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ba20be6 : UNSTABLE in 1h 01m 37s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/502b915 : UNSTABLE in 1h 19m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"548f79459cae6cfcc2a0a9137224e7178accf6e9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-19 13:27:08.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/3/check/tempest-dsvm-full-bionic/f334545/ : SUCCESS in 1h 37m 21s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/3/check/devstack-plugin-ceph-dsvm-bionic/b94e3e1/ : SUCCESS in 1h 46m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a61fd6174920c887351141186f96081ab28a3a30","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-19 13:53:48.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/3/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/3/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"9e60dfb7f6a93712b1a07f6eee9b04613dd38c4a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-19 13:57:32.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-3/ns5_nfs : SUCCESS in 2h 24m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-3/ns5_iscsi : SUCCESS in 1h 36m 47s","accounts_in_message":[],"_revision_number":3},{"id":"d57c3e2d9a269433a8341d4f5a9275ed1063bbc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 13:58:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4227e298dcad4fc9bde5a6ef3bbdb93a : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e9de72308847f5804e227ebdc33a2e : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1a8ebe3c7384b48ab4bf075b386f91c : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a3b797d7c7d4a40bf1ae4191e6277dc : SUCCESS in 11m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0e274e0a26344edb9755624d907e5a9 : FAILURE in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6664d0c0fc2d4ac690e2d43343a4fbce : SUCCESS in 13m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ea209c48e8cf46949b6e92a94d79a0fa : SUCCESS in 1h 02m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/606dd72644924a9da343960a1121b36c : SUCCESS in 1h 21m 42s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1a16f2eaf10042bf8890a38bf68b0211 : SUCCESS in 1h 44m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f30b5468f2d458998009569293ae16c : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce2ccb05b091444bbd75a19831763165 : SUCCESS in 5m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/03c78903d0be4a9685bfbb37008a767f : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/48d70bdeae074f649621a24542901119 : SUCCESS in 43m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bcf9db60763c4905999a33a72265e7c7 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/18a783d2bed5481a8c57b8913eec49c0 : SUCCESS in 1h 05m 53s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c428c4bc36bc43348df5603eab1f4fb7 : SUCCESS in 1h 15m 53s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a8f534f62fc4b489969930de7aceb16 : SUCCESS in 1h 18m 24s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/5bca39faedd64c3eba593e5e83e44140 : SUCCESS in 1h 31m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8dcd4c8fad67453f8bff15c0228e2bbf : SUCCESS in 2h 19m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66bdf3420675440da980ab290832b9c4 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":3},{"id":"ee2a99d191c25d365b9dffcf903b0c270b5c8325","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-19 14:10:12.000000000","message":"Patch Set 3:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3496 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c8b5fa6e58f1ac4616780b8ebcb11da446d14bb7","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-19 14:10:17.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/49/713649/3/check/synology-dsm-cinder-iscsi/ffc6098 : FAILURE in 11m 07s","accounts_in_message":[],"_revision_number":3},{"id":"600699b48a1cbaf2c7deeb84514aa6ae413e36f0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-19 14:50:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/49/713649/3/check/dsvm-tempest-infinibox-fc/5cbbbc9 : SUCCESS in 1h 08m 35s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/49/713649/3/check/dsvm-tempest-infinibox-iscsi/6fce264 : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":3},{"id":"b948edd1dbbc3eb0208d6a2bfecbc2c6d55a5e8d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-19 14:58:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/49/713649/3/upstream-solidfire-check/cinder-solidfire-iSCSI/9f6c1e1/ : SUCCESS in 58m 43s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"1bedc0b4c0933dc070110daaae15129051708be7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-19 15:00:59.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-49-713649-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"f83e7a36f74fbbef6bf2abd274d82ce648624c3d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-03-19 15:22:42.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"d5bf4a8b39824de72cf1723e05c3ae981f240478","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-19 15:30:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f11f3e572e97f5cbf06d98b3f61a5e8e26774653","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-03-19 15:56:37.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/713649-3/Seagate-iSCSI : SUCCESS in 42:53.30\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/713649-3/Seagate-FC : SUCCESS in 42:22.94","accounts_in_message":[],"_revision_number":3},{"id":"4e5dbbcdb39d3043d97b56c606e60730ea00e945","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-19 16:02:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/4/check/fujitsu-eternus-dx-iscsi/f249d94 : FAILURE in 31m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/4/check/fujitsu-eternus-dx-fc/ed5c813 : FAILURE in 31m 57s","accounts_in_message":[],"_revision_number":4},{"id":"2ec86f367791ff7b19d50522628e14f0ed9e72f1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-19 16:57:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/4/iscsi/ : SUCCESS in 1h 18m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/4/smb/ : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":4},{"id":"4debee764bfd557b679e751296351660eb812393","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-19 17:26:39.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-4/ns5_nfs : SUCCESS in 1h 42m 36s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-4/ns5_iscsi : SUCCESS in 1h 56m 1s","accounts_in_message":[],"_revision_number":4},{"id":"793e5eb5f3900a5cd435c7a74be55e57e515a61f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-03-19 18:38:39.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/713649-3/Lenovo-iSCSI : SUCCESS in 50:54.97\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/713649-3/Lenovo-FC : SUCCESS in 50:18.16","accounts_in_message":[],"_revision_number":3},{"id":"e39ba5093cde1acc38bd51572749c6a456d0213f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-19 19:06:10.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"dbb570e6bce369c56cd387b8b361ccbc1c355291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 19:17:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f750f261d6d049a4ae68e8527f1bd412 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b5782dffae41db95d827bebd28998f : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fb785a544cb4711a375a16ba13028fb : SUCCESS in 11m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9081c3564e24f4dae6af20ed53755ac : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c37a2a14d4b40db99988d4add26e186 : FAILURE in 12m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051fb85ac46f43b3ac062401f81daa70 : SUCCESS in 17m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/40c50a5ffe3749c8bcb041f6fc552969 : SUCCESS in 1h 01m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dfc942ec823340bb9bc73785f0b3604e : SUCCESS in 1h 23m 56s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1836f7eca2d04e16ae501e5ecebf2620 : SUCCESS in 1h 29m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb988a038d404f3b9d45adb4bf921244 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/371030eb10ed43b8be5eb73922731b6a : SUCCESS in 6m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fce7e5dbe16340bbbaedf61c1b6966e5 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/40e10a732471480a8c6638d8442c5dfd : SUCCESS in 48m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f6c4207f7b0b486ba9b3c20ce8fbe9c5 : FAILURE in 8m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/84ccb6cbd2dc4906add31143adfec1dc : SUCCESS in 1h 16m 57s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a71fc4b97a6f491f97e0065cce9b76f8 : SUCCESS in 1h 24m 54s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/20eb4029fa254ebb8c8c7ac5be4684b6 : SUCCESS in 1h 09m 00s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/bdc24e16d4944adcad95238df65e3e9b : SUCCESS in 1h 13m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/350893fb52a345749509d7c2f56d87fb : SUCCESS in 2h 34m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07fa608fa24c4216bc353282b2f8e83c : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":4},{"id":"f60f499ca8a0fc408601da74f4d63672cbdd53a9","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-19 19:23:06.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/d3c776e : UNSTABLE in 1h 04m 04s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/01dd490 : FAILURE in 1h 58m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"e34e36c83141cd7359aef07f08d25e0925e4a240","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-19 19:29:04.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/49/713649/4/check/synology-dsm-cinder-iscsi/52c4b67 : FAILURE in 10m 35s","accounts_in_message":[],"_revision_number":4},{"id":"f56da981fcfa884850c5f0ec9893c5084e8f8a6f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-19 20:19:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/49/713649/4/upstream-solidfire-check/cinder-solidfire-iSCSI/309854d/ : SUCCESS in 59m 21s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"75247e8d1543704140294bc90e604fa8cdbf2325","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-19 20:34:07.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/4/check/tempest-dsvm-full-bionic/981ad34/ : SUCCESS in 1h 47m 27s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/4/check/devstack-plugin-ceph-dsvm-bionic/3b42e70/ : SUCCESS in 1h 52m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c902965241aef1983937c44a69ec5e46515200f8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-19 20:37:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/49/713649/4/upstream-check/cinder-cDOT-iSCSI/2e5a6d7/ : SUCCESS in 48m 46s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/49/713649/4/upstream-check/cinder-cDOT-iSCSI-IPv6/e11dce8/ : SUCCESS in 1h 18m 36s (non-voting)\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/49/713649/4/upstream-check/cinder-cDOT-NFS/313283b/ : SUCCESS in 46m 58s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/49/713649/4/upstream-check/cinder-cDOT-NFS-IPv6/8cacde4/ : SUCCESS in 1h 14m 46s (non-voting)\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/49/713649/4/upstream-check/cinder-cDOT-FCP/7846a3b/ : SUCCESS in 59m 41s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"4d84aa1f23624660e130c21954f8ecf9666e598d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-19 20:40:03.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-49-713649-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"a8afc530d82bda6e1102c993e9ee7626d6740148","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-19 21:23:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/49/713649/4/check/dsvm-tempest-infinibox-fc/a548bb7 : SUCCESS in 1h 13m 02s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/49/713649/4/check/dsvm-tempest-infinibox-iscsi/26bcc2a : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":4},{"id":"148072b13e3be1bebd5f9e489d4be791e9c15a8d","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-19 21:39:58.000000000","message":"Patch Set 4:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3501 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a41c0d537a2bfc5a75e463b650417a4eedbdaded","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-19 22:54:01.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d3eb47728496b423b82d0f6ca986d694d33f420f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-03-19 23:31:50.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"bc275b617a3c473d42f8e5f875e191fd537710df","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-20 01:43:09.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/4/check-not-vote/ext-cinder-zuul/44b4c4c : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":4},{"id":"48b44c72f0c6df6dc2facd80c1472c1ba19c577c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-20 02:29:20.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/49/713649/4/check/DellEMC_VxFlexOS/ded2d0d/DellEMC_VxFlexOS/463 : SUCCESS in 2h 13m 08s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"aec89863c1f2421221494d2d91ba8d963d7a14b5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-20 03:35:21.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/49/713649/4/check/DellEMC_Unity_FC/299be05/DellEMC_Unity_FC/713 : SUCCESS in 1h 36m 01s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/49/713649/4/check/DellEMC_Unity_iSCSI/5d33798/DellEMC_Unity_iSCSI/692 : SUCCESS in 1h 51m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c4aa35e5ec9a21ad581cded4adfe0ce57487e698","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-20 03:50:55.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"ef212b70982a02b1d6f39aeadf9c38b8ff5f3208","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-03-20 06:58:15.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/49/713649/4/check/DellEMC_SC_FC/040bb0f/DellEMC_SC_FC/457 : FAILURE in 1h 08m 44s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/49/713649/4/check/DellEMC_SC_iSCSI/ac175da/DellEMC_SC_iSCSI/457 : FAILURE in 2h 50m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"d4b485bc3366a322e08fb009c5da911284673281","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-20 07:44:04.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"0460de7815ffec71bf9a9a64ddd64ff440e65caa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-20 08:11:35.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"6901b16f257f170532faa017d48d560144e6cfb4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-20 08:34:45.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"6a5228ca017efeafa9c50381dfeb063624b83cd1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-20 10:17:13.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/4/check/DellEMC_XtremIO_FC/27d75e3/DellEMC_XtremIO_FC/807 : SUCCESS in 1h 29m 27s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/4/check/DellEMC_XtremIO_iSCSI/a146747/DellEMC_XtremIO_iSCSI/754 : FAILURE in 1h 59m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"f73faccceec383d1f865d466dfa0e08566efc782","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-20 11:28:52.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/4/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"54b6222ac0e26ba75b5cb773fc77ee0c502574d8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-03-20 13:52:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/4/20200320202751/NEC-ISCSI : SUCCESS in 2h 16m 28s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/4/20200320202751/NEC-FC : SUCCESS in 2h 24m 44s","accounts_in_message":[],"_revision_number":4},{"id":"35d48679b1c416553c775b558256bc77ca120f68","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-20 20:02:02.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"8c0f962fdc38d1eeba46b0645bec549da4c2885d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-03-20 20:03:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/49/713649/4/check/cinder-storpool-tempest/1b77d68/ : SUCCESS in 3h 00m 48s","accounts_in_message":[],"_revision_number":4},{"id":"3689a6ff9b6d2bc22c812c831503d78bb60f51a8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-20 21:03:32.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"f43942ac3f27b4d38972dc5bf03945547e5f8c2e","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-03-21 09:21:20.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/490/index.html : FAILURE in 35m 07s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/636/index.html : FAILURE in 40m 53s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/676/index.html : FAILURE in 35m 33s","accounts_in_message":[],"_revision_number":4},{"id":"b540bb57b79685707e7abf338d5acc0788790d31","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-03-21 18:28:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/49/713649/4/check/nimble-iscsi-driver-dsvm-volume/cd6ae98 : FAILURE in 2h 19m 23s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/49/713649/4/check/nimble-fc-driver-dsvm-volume/c5f402a : FAILURE in 20m 34s","accounts_in_message":[],"_revision_number":4},{"id":"05a235259e6929531ed5f106d37030dec76e609f","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-03-22 15:36:29.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/49/713649/4/check/kaminario-dsvm-tempest-full-iscsi/5ed5c37 : FAILURE in 3h 09m 45s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/49/713649/4/check/kaminario-dsvm-tempest-full-FC/9b6756b : FAILURE in 3h 08m 34s","accounts_in_message":[],"_revision_number":4},{"id":"6633c6aa701cee07900e8e803076c20236478817","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-03-24 06:35:21.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/49/713649/3/check/kaminario-dsvm-tempest-full-iscsi/8d70e39 : FAILURE in 5h 00m 17s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/49/713649/3/check/kaminario-dsvm-tempest-full-FC/c1e9cc7 : FAILURE in 4h 05m 31s","accounts_in_message":[],"_revision_number":3},{"id":"1c2b3e1e2d72f9b09efde27df5a6dba5bc052930","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-03-25 12:21:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"84b51eb067a4bfec69d6faf884bd39f72a8a60aa","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-25 13:05:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"de367873f5bf9cf33af80e81ca535dffee51e0b6","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-25 13:06:37.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f6e66e9f274f43f7fa141afd488fb31b29a2f809","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-25 13:39:20.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-tgtadm : FAILURE\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-lioadm : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"05b79bc159255ee5bd79fee0d6ef5ce8d0d9b616","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-25 14:09:51.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/5/iscsi/ : FAILURE in 1h 03m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/5/smb/ : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":5},{"id":"0e090a7b7556b1512b6c9a6e360b5dec71272340","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-03-25 14:48:27.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1690 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1724 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1896 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"76131262cb2c46a5c01f06f06520713e95e27182","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-03-25 15:08:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/5/check/fujitsu-eternus-dx-iscsi/c4a866c : FAILURE in 31m 40s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/5/check/fujitsu-eternus-dx-fc/163fec6 : FAILURE in 31m 01s","accounts_in_message":[],"_revision_number":5},{"id":"c165d91a5cc975fbbc18dee4a6e8fbe9e0b71e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 15:46:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6970f5f20d114e44ae9f9f84f5104aa7 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/203817ecd8934d00a12253acca38b284 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d00546e0334648adaa0b6c6d98546cbe : SUCCESS in 12m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bf70067e4ba4b699085103a4409be04 : SUCCESS in 11m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c57cfd395a254eb190cd36794abb075e : FAILURE in 10m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dccf5d8437f443ce8a8dc31f0d0b888c : SUCCESS in 18m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/731f3f34fc07442ba152e278b8b3ded9 : FAILURE in 47m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30ab553233be4d18ade93d12c3d7dac2 : SUCCESS in 1h 38m 33s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/19a6f536385a4efcb627b2c718cf7ee4 : FAILURE in 1h 40m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a0dba4579dd47689e075487045eec46 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f37096ddefda4f94af100d8df9df6a83 : SUCCESS in 5m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/77dfcfb40c1d40d4ac55a3561f874081 : SUCCESS in 9m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/172aaf92513b446c84bc60662ec9d611 : SUCCESS in 42m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14ff45ad9c634b03b8187762c347680a : FAILURE in 4m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/517677efb12d400eaf4dda91088600af : SUCCESS in 1h 22m 08s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b8dce6b935b94afbbbbadce8b90a5f5a : FAILURE in 1h 45m 36s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9d79c8652d934984972b192892b5923d : SUCCESS in 1h 27m 23s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/c8c95011eb4549d78da8e23567541865 : SUCCESS in 1h 45m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fd990fdf7e643eea8756d099ebf5ee8 : SUCCESS in 2h 34m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9de7412ca264478793825cccd31b1ef9 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":5},{"id":"4ac5362e61ea70fa32b173eae8fcfdd042ab11a8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-25 16:23:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/5/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/a56c860 : UNSTABLE in 1h 04m 21s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/5/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/7de5b95 : UNSTABLE in 1h 15m 13s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"828711ea5e4f0ba12180664a7c73a8d7541243a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-25 16:30:50.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"2381c22e256a4576cf08b4444a2c65f45efc2eb9","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-25 16:37:05.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"50de958c90bb26c5b449657a76cd2b41356c6c62","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-25 18:02:30.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/5/iscsi/ : FAILURE in 1h 24m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/5/smb/ : FAILURE in 27m 33s","accounts_in_message":[],"_revision_number":5},{"id":"bf6d04d97c66aef500f54580889fd374b15783b8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-25 18:25:14.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-5/ns5_nfs : SUCCESS in 2h 3m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-5/ns5_iscsi : SUCCESS in 2h 6m 6s","accounts_in_message":[],"_revision_number":5},{"id":"f47439228c8481e42d90d06d044df46c95d9ce9e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-25 19:43:30.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/5/check/tempest-dsvm-full-bionic/ae29af0/ : SUCCESS in 1h 42m 25s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/5/check/devstack-plugin-ceph-dsvm-bionic/92ba691/ : SUCCESS in 1h 48m 42s","accounts_in_message":[],"_revision_number":5},{"id":"267b80c9e7c98e8735ee871d70788ab0325f7e31","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-25 20:09:04.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"21e011c049f95a4f51681912f78f48c3010c75f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:04:08.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d25ee21cee98440b9f1fb90341741424 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4289716f5374f0d841713dafe1a177a : SUCCESS in 10m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/435b5b2f1fe84ea2bc56e18f279daa62 : SUCCESS in 9m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5f99c30d6064d5bafc270eb0e0c3acf : SUCCESS in 19m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a9f2eae262e4ef89df6f152ae06f5dc : FAILURE in 17m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d2a07aa6e64e1a9a762bdf45206ab5 : SUCCESS in 22m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e19a65eb344e48bfa60a5de3c8cb8088 : SUCCESS in 1h 12m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/881a98d4d3f243a7b867f6c1840de1f9 : SUCCESS in 1h 26m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dbb5391e40174b84aaed478a268510d8 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32fad906a9d94c46af054ddb1fdb6e65 : SUCCESS in 7m 06s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/121b1cf85e444065b6527cd6aff75468 : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2caa89aa57be41bd81c7f15c075ba81d : SUCCESS in 39m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/15fec18a20d2418f81a7b3c005381219 : FAILURE in 9m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9705bd225f994a67975bf775a5f00a7e : SUCCESS in 1h 22m 31s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a2fdf84eeea04e71aa716513e6858031 : SUCCESS in 1h 40m 44s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4366fb9537804c27884633308a50ca54 : FAILURE in 1h 05m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bb33af0ec75f43adb98ef65f1bc3fb57 : SUCCESS in 1h 30m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/43c17d79854b40729842c20c7b9fd70d : SUCCESS in 1h 36m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ee4f15ecb3b4ac08a9213c861bc2699 : SUCCESS in 2h 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af2c792a594c470f8aec3511f9a24f8a : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":5},{"id":"6aaf2b3afb9e965ef6548c57e4c844c9c7c13c36","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-25 21:43:22.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/5/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/242ac3b : UNSTABLE in 1h 01m 33s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/5/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/e279289 : UNSTABLE in 1h 29m 35s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"135a7b7a99e32a3da5a2c4db715c7d7015589ad1","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-03-26 03:50:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/5/20200326102353/NEC-FC : SUCCESS in 2h 12m 08s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/5/20200326102353/NEC-ISCSI : SUCCESS in 2h 26m 17s","accounts_in_message":[],"_revision_number":5},{"id":"f91a78644a9f5d95932f90aaa3ae2c20a8dd34a5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-26 05:50:49.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/5/check-not-vote/ext-cinder-zuul/3abc68e : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":5},{"id":"c0065b52b2033b2c4f0bff54c707cd9d6baa1825","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-26 07:47:44.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"a9774575c7a2903f025f995496ed58ae24de3282","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-26 08:27:41.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/49/713649/5/check/DellEMC_Unity_FC/399c24a/DellEMC_Unity_FC/759 : SUCCESS in 1h 24m 28s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/49/713649/5/check/DellEMC_Unity_iSCSI/44ffaa0/DellEMC_Unity_iSCSI/736 : SUCCESS in 1h 58m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"19d99719cc3470576532bb069c18ecde53a46d0a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-26 09:08:37.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"491aa7c09e3cfc987ce7a4bc701a4be09d8818eb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-26 09:41:13.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"d3dac3d6fd0651242342305ffc9a5c2a1fa17689","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-03-26 10:09:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/5/iscsi/ : SUCCESS in 1h 00m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/5/smb/ : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":5},{"id":"62e6f10000a8ebd966432c7cd8de208a7a1d8431","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-26 11:06:31.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/5/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/cdfde08 : FAILURE in 1h 57m 27s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/5/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f5fa946 : UNSTABLE in 1h 24m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"ad84d86e4427112efe5ada3230194e9f9d0f9b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 11:26:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfadf78454e74b899f95b2cfaa798a2d : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66037df99554e38a9b98dbc19471351 : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71c08f9e62654cda9526289f2a445a54 : SUCCESS in 14m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3119ff4d8fd14256bac7ff14d40133e2 : SUCCESS in 16m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e11fa23f02294a5c95b23bc99eb5c4e0 : FAILURE in 15m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d320c635bf214f07be9bfdd575ab25b6 : SUCCESS in 16m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d142213d3ffd4e839375e61a4c2c0212 : SUCCESS in 1h 05m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5686ab07022f4cdf90b0de8271024313 : SUCCESS in 1h 28m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/65e684a8c2014b5e93f18a343c2ea745 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ebcb75bcb1b4fb29a1fd4fd454c7851 : SUCCESS in 8m 11s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d6592b9c9e4047fd9c997150a55a721b : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d3519d5cebb748fba24abdd8891ed56e : SUCCESS in 45m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/91c76fee23824ede9833c84a7d2495d0 : FAILURE in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/50855b0c9a0546f7b63afedb5a4dd70a : SUCCESS in 1h 10m 13s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4e945c547fc4434f87e554aff792bc7a : SUCCESS in 1h 22m 49s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ddb2751cfb274384ae87aa6a71437933 : SUCCESS in 1h 16m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6296004f56f14d219a72267b96462991 : SUCCESS in 1h 52m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fa06d60694af407abf61356bf20eb153 : SUCCESS in 1h 56m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f062f513c284836a78815dcf69afddb : SUCCESS in 2h 16m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86fabd33adb2415598de9ef5576416f2 : SUCCESS in 1h 27m 55s","accounts_in_message":[],"_revision_number":5},{"id":"2a76a06d9b38d9c15928c8b6e384b23418671428","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-26 11:29:22.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/49/713649/5/upstream-solidfire-check/cinder-solidfire-iSCSI/3810902/ : FAILURE in 1m 10s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"9dce9229ad1d1e2ccf59f78f2e6461698fb6cfb8","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2020-03-26 11:38:29.000000000","message":"Patch Set 5:\n\nBuild Failed. Comment recheck linstor to rerun the CI. \n\n* linstor_ci http://linstor-openstack-ci.linbit.com/logs/3538 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"811b3f283df8e8f0392221c6d5dc67281e868e45","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-26 11:38:54.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"06190c429097121cf815217d1ec73e59c0653805","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-03-26 12:23:51.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/49/713649/5/check/synology-dsm-cinder-iscsi/50b224c : UNSTABLE in 56m 28s","accounts_in_message":[],"_revision_number":5},{"id":"73216e835d083d55f148fe1a1b7a8a057e274ef1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-26 12:46:08.000000000","message":"Patch Set 5:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-49-713649-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"a64864f9246a1ee3e1ffbd3fe0ce1f937fe1c2b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-26 13:02:00.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"64a2156dca757de113f450e41f3c1814f604c587","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-03-26 13:27:29.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/713649-5/Seagate-iSCSI : SUCCESS in 45:52.43\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/713649-5/Seagate-FC : SUCCESS in 41:22.38","accounts_in_message":[],"_revision_number":5},{"id":"546bd33f3c80319ff81bbd83084161ce46052f0f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-03-26 13:32:44.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/49/713649/5/check/DellEMC_SC_FC/7586b06/DellEMC_SC_FC/481 : SUCCESS in 1h 14m 01s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/49/713649/5/check/DellEMC_SC_iSCSI/67d3b35/DellEMC_SC_iSCSI/481 : SUCCESS in 2h 05m 35s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"0d9697c2f9ce02c5db35d46cdabbb36a712a6fb4","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-03-26 13:38:03.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/713649-5/Lenovo-iSCSI : SUCCESS in 50:51.07\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/713649-5/Lenovo-FC : SUCCESS in 47:12.34","accounts_in_message":[],"_revision_number":5},{"id":"00f464058e83af26fa97de5017dd361472e0319f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-26 13:43:24.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/49/713649/5/check/DellEMC_VxFlexOS/7c3dd5f/DellEMC_VxFlexOS/486 : SUCCESS in 2h 16m 14s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"13ce3c4115b4162de884dd4f26f9a138f219a449","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-03-26 14:34:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/535/index.html : SUCCESS in 1h 17m 01s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/681/index.html : SUCCESS in 1h 37m 51s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/723/index.html : SUCCESS in 1h 22m 10s","accounts_in_message":[],"_revision_number":5},{"id":"8a1758953c35a89a3256e3beda6aadf71fa8dfab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-26 15:28:44.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"67789a843901e0531de0534a395034757b52859f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-27 04:59:45.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/49/713649/5/check/DellEMC_Unity_FC/0c38ce9/DellEMC_Unity_FC/769 : SUCCESS in 1h 21m 58s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/49/713649/5/check/DellEMC_Unity_iSCSI/cbad9b8/DellEMC_Unity_iSCSI/745 : SUCCESS in 1h 57m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"be7b7a9e73fdfa02dde88ce558652612cba39662","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-27 09:05:52.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/5/check-not-vote/ext-cinder-zuul/d166749 : FAILURE in 26m 01s","accounts_in_message":[],"_revision_number":5},{"id":"9ecd35784dbcd1c89ea2bac7ffcba26754c9734f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-03-27 09:37:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/49/713649/5/check/cinder-storpool-tempest/57734a3/ : SUCCESS in 3h 03m 00s","accounts_in_message":[],"_revision_number":5},{"id":"9cca8025432e75750eed2fc7809ee9f50d2af098","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-27 10:25:56.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-tgtadm : SUCCESS\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"07c054ceffb3d06bcacd1ddb27d16c9074c2d11c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-27 12:56:11.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"548c466e1e39c6592711bfcc6f726125d2ed07e6","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-27 19:02:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/49/713649/5/check/DellEMC_PowerMAX_FC/a224b45/DellEMC_PowerMAX_FC/541 : SUCCESS in 6h 01m 14s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/49/713649/5/check/DellEMC_PowerMAX_iSCSI/e7dfe1f/DellEMC_PowerMAX_iSCSI/449 : SUCCESS in 7h 58m 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"298b7e1c39a1ee679727e87efc9f0c33f4c1390c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-03-27 19:36:51.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/5/check/DellEMC_XtremIO_FC/4f17dee/DellEMC_XtremIO_FC/851 : SUCCESS in 1h 25m 05s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/5/check/DellEMC_XtremIO_iSCSI/75ed132/DellEMC_XtremIO_iSCSI/793 : FAILURE in 5h 09m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"cf0267418c6edf77eaf055b55c553ea8cad3c6d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-27 20:31:43.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-tgtadm : FAILURE\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/49/713649/5/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"74efd5d32fb26927af2fc0069b0f47a1375d0ce1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-03-29 14:18:28.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"240592de60978594bc70b687566865842098bcb5","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-12 20:50:44.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"f0bed2bf0b1fc18c71d9b78d12105af8b64564da","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-12 22:07:55.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"a1c1b7d90de6a739840b4c29b742f97bf3f6b58f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-14 13:23:44.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks like Gorka\u0027s comments have been addressed. Would love to see his review on the latest.","accounts_in_message":[],"_revision_number":5},{"id":"b3446945fd1afc2998a8135a8637a9a79f83f167","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-04-14 13:37:19.000000000","message":"Patch Set 5: Code-Review+1 Workflow+1\n\nThe code looks great, but there are no unit tests...","accounts_in_message":[],"_revision_number":5},{"id":"989cdb022502fca725dd72d7e3dcdf7297e61409","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-04-14 13:37:29.000000000","message":"Patch Set 5: -Workflow","accounts_in_message":[],"_revision_number":5},{"id":"d015b083b84f03cb287601773017ac5315a2ae59","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-14 13:37:29.000000000","message":"Removed Code-Review+2 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":5},{"id":"6f501c9bd0e04b4020812f287e82ce973b930b15","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-24 15:53:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ab0b15e5379216a9a56b26f6248eabdc76c53528","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-24 16:33:12.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/6/check-not-vote/ext-cinder-zuul/daa951b : SUCCESS in 39m 03s","accounts_in_message":[],"_revision_number":6},{"id":"39cb9401437c74960759ea06f1b6891bc43751db","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-24 16:45:34.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"6561f53687c9d6e146d9ef7f9203ea0dd1aa4b9c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-24 16:49:13.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/6/iscsi/ : FAILURE in 53m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/6/smb/ : FAILURE in 55m 07s","accounts_in_message":[],"_revision_number":6},{"id":"b3d6e2e37be41598665a3ab716be417cc2389e2a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-24 16:51:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/49/713649/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/fced785/fced785.html : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":6},{"id":"8e4c2509fde9aacfb662c72d16b7c1232203e100","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-24 17:59:20.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-6/ns5_nfs : SUCCESS in 2h 5m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-6/ns5_iscsi : SUCCESS in 2h 4m 19s","accounts_in_message":[],"_revision_number":6},{"id":"2586eadab7963f9cfdb96d1bead3919e9421cf7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:13:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/237479e7cf0748849ecdd75f0ae06cc5 : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba130333f7814cfbb11eb67f3e1a8ca6 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3276882494f048c4a34ae43cf70f442d : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/061bd3ade91345c2958428faaaf21670 : SUCCESS in 13m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29af0a61b5064ed8a0f94dc19626ae63 : SUCCESS in 12m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/257c2b5780cf4bf09ca52ceed37c8093 : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1dc47eb3b734c008ca99da266742ad4 : SUCCESS in 17m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4645faa9499a44be87ccf245e330c4cc : SUCCESS in 5m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4cc8afc59e874d4b8ae45fe3f26a374d : SUCCESS in 1h 17m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c845c0d8af74484be257e6ce3ea8637 : SUCCESS in 1h 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c03b4790809041e9bc3ea2a4e552e72b : SUCCESS in 27m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/697e02187f024745a4793a8db751f881 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c15b134c5a2449098485cec2bff97b0f : SUCCESS in 8m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3bf2e949de964d8e836f6b67f1c0a3c5 : SUCCESS in 8m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e7e9b46e64f64764ae3f9452b6a47c10 : SUCCESS in 47m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96c58321397b420e815eacd997bc6c75 : FAILURE in 8m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/43b904dad07248dda4b91d624daedac1 : SUCCESS in 1h 20m 04s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9192e44249db4ccbb4460b3d0ae44177 : TIMED_OUT in 2h 07m 13s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dbb52cacb6ed4243a9b3463385b91901 : SUCCESS in 1h 22m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2466d2e790154053a213f38d856dfaca : SUCCESS in 1h 18m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/32df0829152b4f6ba5c5bf5d9a2d4617 : SUCCESS in 1h 12m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e83b9e77972457dab66e02974e5834d : SUCCESS in 2h 09m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdbdcac4bf004b86b6db1a37c60b17ea : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":6},{"id":"c4e273b1f6d0b6b6236e2cfb8ebbc5bf7b0e69e5","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-24 19:19:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/6/check/fujitsu-eternus-dx-iscsi/4fd8365 : FAILURE in 35m 47s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/6/check/fujitsu-eternus-dx-fc/314f321 : FAILURE in 37m 08s","accounts_in_message":[],"_revision_number":6},{"id":"ee57fa7f23d8928ee1b44707c8646b5394093260","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-24 20:41:58.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b50108d816e69e5ac10a567c935abf5159366153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 20:43:09.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"e0e5862637acdc7cbda36e177196bbd8ece91b96","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-24 21:19:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b421fcf : SUCCESS in 54m 30s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/d25069d : SUCCESS in 1h 05m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"a94a9018f64fd33c20d410b8529886aadaadbc8a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-24 21:20:06.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/6/check-not-vote/ext-cinder-zuul/86c5b96 : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":6},{"id":"14fbab8574f1815bf728ac040446502528ab289e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-24 21:27:55.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/6/check/tempest-dsvm-full-bionic/7902c9d/ : SUCCESS in 1h 54m 43s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/6/check/devstack-plugin-ceph-dsvm-bionic/2d0c70d/ : SUCCESS in 2h 10m 26s","accounts_in_message":[],"_revision_number":6},{"id":"d84aed589d2939450cd112c0dc9f02ba0b453a3d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-24 21:34:30.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"5f9e0bae1fa19a9c4e92ed58f4cc88e531523b46","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-24 21:40:04.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/6/iscsi/ : FAILURE in 57m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/6/smb/ : FAILURE in 56m 37s","accounts_in_message":[],"_revision_number":6},{"id":"53a9a2e547d088ab0c99705158fc56214724936f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-24 21:42:28.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"66e5d5e0a930e6c4513f8247f6f782bc260076c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-24 21:51:28.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/119/index.html : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"3464d76cb587cf8ca441404c239e8fd325c7616c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-24 22:21:38.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/7/check/fujitsu-eternus-dx-iscsi/487beb0 : FAILURE in 38m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/7/check/fujitsu-eternus-dx-fc/93e3f5d : FAILURE in 38m 24s","accounts_in_message":[],"_revision_number":7},{"id":"648fb2fa9175b18d7cc2db4ca05faaf79a959071","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-24 22:23:38.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/7/check-not-vote/ext-cinder-zuul/c1dee01 : SUCCESS in 40m 33s","accounts_in_message":[],"_revision_number":7},{"id":"cf5924097724b9ca5d0dc015fa78f4165201412d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-24 22:34:48.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"b7b13813610360f1ee24bd8abf64e00138416301","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-24 22:40:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/49/713649/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/108270a/108270a.html : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":7},{"id":"aac3a1e70fb8a033e043831f5f02afeff5d82841","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-24 22:51:21.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/7/iscsi/ : FAILURE in 54m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/7/smb/ : FAILURE in 1h 08m 25s","accounts_in_message":[],"_revision_number":7},{"id":"2de5471a378258ad98d8c9dd85a0dc8020fc3327","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-24 23:35:09.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-7/ns5_nfs : SUCCESS in 1h 51m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-7/ns5_iscsi : SUCCESS in 1h 52m 15s","accounts_in_message":[],"_revision_number":7},{"id":"c652ac2d4ff65c4b9842821c50d006ea314ad56a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-25 00:20:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/d96f39e : SUCCESS in 58m 41s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f0294bc : SUCCESS in 1h 23m 42s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"30b3df6476b7da2df8e037cff00be2464e7666ec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-25 00:30:26.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/7/check/tempest-dsvm-full-bionic/2277210/ : SUCCESS in 1h 48m 27s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/7/check/devstack-plugin-ceph-dsvm-bionic/095f5ce/ : SUCCESS in 1h 43m 52s","accounts_in_message":[],"_revision_number":7},{"id":"c166a6f2045ea9e0eeb40d5c99b13d2a3e970a2b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-25 02:59:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/49/713649/7/check/dsvm-tempest-cisco-zonemanager-job/8b792af/ : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":7},{"id":"1b17f824e67a5b9c3cd4a3fab36c57dcae0b145a","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-25 03:19:04.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/49/713649/7/check/nimble-iscsi-driver-dsvm-volume/b8197a9 : FAILURE in 3h 32m 24s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/49/713649/7/check/nimble-fc-driver-dsvm-volume/a14b066 : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":7},{"id":"1768809582d27991b7f5bfc2811ebc53487b0d8d","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-25 08:37:17.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"bf0b880f84e5f4c9f35671903b00783d038bdcb9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-25 09:17:04.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/7/check-not-vote/ext-cinder-zuul/8d00f62 : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":7},{"id":"84b5b86d6dbeaeb075ec548cd318811b9a385219","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-25 09:29:19.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"f502f2143d99ae57ea13a6d9164a0f0e7e4728e9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-25 09:44:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/49/713649/7/check/dsvm-tempest-cisco-zonemanager-job/9fbcc17/ : FAILURE in 1h 01m 43s","accounts_in_message":[],"_revision_number":7},{"id":"0ac3ae950f4c4039f31136c8f06cbe01170bc150","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-25 09:51:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3e57063 : SUCCESS in 56m 48s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ff91761 : SUCCESS in 1h 13m 30s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"49087c17319659407f441b38f4ce5e1f6a82833a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-25 09:54:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/7/iscsi/ : SUCCESS in 1h 16m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/7/smb/ : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":7},{"id":"ebf9e834ad3953ff681781630fda685c92a971b8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-25 10:29:09.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/7/check/tempest-dsvm-full-bionic/d93f210/ : SUCCESS in 1h 28m 15s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/7/check/devstack-plugin-ceph-dsvm-bionic/5691d7c/ : FAILURE in 1h 44m 55s","accounts_in_message":[],"_revision_number":7},{"id":"4803d22d39aee4b8fbd5fdcb1559c19f00c005ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 11:17:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/67c72bdbafff404c9661fcb2aa5516b3 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7c89777e8c34d449e8881bc5ccabcb6 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541dd782cbde4e18a7091096b9ed699b : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01a744f0b45a4631afc2c213e38abcf1 : SUCCESS in 11m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/274740bcc62d4a8584905a6682b56c37 : SUCCESS in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca3adcb73c3a48d0bb646e2e0d0b7d4f : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac48e18e9a754981a8f24f686d9cf139 : SUCCESS in 13m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c20d39054131462eacad716b438d2f04 : SUCCESS in 3m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/254c4e44264d48389cb46aeffe810762 : SUCCESS in 1h 07m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7afd853a09d9402f90eb1c6be30740fc : SUCCESS in 1h 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88d9693b2667480dbedaa8d8db170275 : SUCCESS in 23m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b90cb63887284f709bc170e0795f01c1 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b032128f0634f29a4c068e2d9812898 : SUCCESS in 6m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3f1149077751412784b0e1ffbe797443 : SUCCESS in 9m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b18895b9e55441caadfc44fd2ea1e4a : SUCCESS in 38m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06bd88db5587457e9692fad055bbbc49 : FAILURE in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d10b603fb4964fd99966740a32bf9938 : SUCCESS in 1h 02m 14s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2d01dded93c2449f8baaa82e96d1ebd3 : SUCCESS in 1h 47m 53s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1dd72c96d47e4b56b5e8a7e1594bbbe7 : SUCCESS in 1h 40m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/134143e04a45479d816b7fddcda23222 : SUCCESS in 1h 15m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/da219ac7eade4b369a2f02e531e95440 : SUCCESS in 1h 17m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23741c550c7a4e39a5dc46e0f896b8f8 : SUCCESS in 2h 33m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77fe518c5a9b4a1bbed54b62de177847 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":7},{"id":"d6a9cf19f2235102309f4220695b5168ed670a43","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-04-25 11:21:12.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/49/713649/7/upstream-solidfire-check/cinder-solidfire-iSCSI/87f7627/ : FAILURE in 2m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"18900fc872439489be4a11d37e9fce6e9efe7281","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-04-25 11:27:07.000000000","message":"Patch Set 7:\n\n* datera-dsvm-full http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/refs-changes-49-713649-7/index.html : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"0a45f014fbe2b448cdc382760842db11b834d832","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-25 12:21:29.000000000","message":"Patch Set 7:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-49-713649-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"d2e3df7edf0ffba94ddcaa6f8d3b0291a26afae8","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-04-25 12:27:41.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/713649-7/Lenovo-iSCSI : FAILURE in 15:53.80\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/713649-7/Lenovo-FC : FAILURE in 23:00.36","accounts_in_message":[],"_revision_number":7},{"id":"9ece6059b67d3deada3d15f63a71fa0900219463","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-25 12:44:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/49/713649/7/upstream-check/cinder-cDOT-iSCSI/b50faaf/ : SUCCESS in 44m 59s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/49/713649/7/upstream-check/cinder-cDOT-iSCSI-IPv6/cbea422/ : FAILURE in 1h 13m 48s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/49/713649/7/upstream-check/cinder-cDOT-NFS/4a83a1f/ : FAILURE in 4m 02s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/49/713649/7/upstream-check/cinder-cDOT-NFS-IPv6/778e1e3/ : SUCCESS in 1h 13m 59s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/49/713649/7/upstream-check/cinder-cDOT-FCP/36d7031/ : FAILURE in 1h 26m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"1fac68ded08fe96d5b4df7b973a2b0f39efacb61","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-04-25 13:05:14.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/713649-7/Seagate-iSCSI : SUCCESS in 41:05.71\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/713649-7/Seagate-FC : SUCCESS in 37:12.37","accounts_in_message":[],"_revision_number":7},{"id":"86bc4f876af6d9dec0fc246b68f3576023a8a273","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-25 21:49:02.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/49/713649/7/check/DellEMC_Unity_FC/8de8197/DellEMC_Unity_FC/1071 : SUCCESS in 1h 15m 05s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/49/713649/7/check/DellEMC_Unity_iSCSI/c8f1517/DellEMC_Unity_iSCSI/1020 : SUCCESS in 1h 46m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"f09d62a2f3b4256f7a5360bbc60b9299454af4f0","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-04-26 02:54:59.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/49/713649/7/check/DellEMC_SC_FC/9e111b8/DellEMC_SC_FC/693 : SUCCESS in 1h 19m 12s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/49/713649/7/check/DellEMC_SC_iSCSI/78394f1/DellEMC_SC_iSCSI/692 : SUCCESS in 2h 14m 07s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"70e3b44a624929e210ea773424cbf013426cacc9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-26 03:05:18.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/49/713649/7/check/DellEMC_VxFlexOS/7211c8a/DellEMC_VxFlexOS/701 : SUCCESS in 2h 27m 10s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"f87e977d4717cb6a0c992ac6f7fd5e234207036d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-26 04:05:20.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/49/713649/7/check/DellEMC_VNX_iSCSI/efcd928/DellEMC_VNX_iSCSI/605 : SUCCESS in 2h 34m 47s\n- DellEMC_VNX_FC http://publiclogs.emc.com/49/713649/7/check/DellEMC_VNX_FC/55f5aef/DellEMC_VNX_FC/607 : SUCCESS in 59m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"3095677e1267b849cfa0d05e54a696ff12ba10bc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-26 08:33:09.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/7/check/DellEMC_XtremIO_FC/0cf7fb0/DellEMC_XtremIO_FC/1174 : SUCCESS in 1h 30m 57s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/7/check/DellEMC_XtremIO_iSCSI/93e63df/DellEMC_XtremIO_iSCSI/1065 : FAILURE in 2h 14m 10s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"0340b84983dab47c3924be4805892b78d1dbd381","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-26 14:04:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/842/index.html : SUCCESS in 1h 19m 42s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/991/index.html : SUCCESS in 1h 45m 51s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1030/index.html : SUCCESS in 1h 23m 07s","accounts_in_message":[],"_revision_number":7},{"id":"0bf63c08b069f958d02c08c0d1fbc0b135d25e06","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 06:41:19.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"9e6ecb154f968fe66d751cc71382ea6c10278b5a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 07:34:52.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"d77bae7bf1f4ed84b5a7d11bae1e1a7b00f0595c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 14:04:55.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"8452cd7c7c54ba78e718c9e1fca68af79f84b28d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 14:53:56.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"e21321f74695f564485bfeae292935ba9b784c28","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-29 14:08:03.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"1a0e82e162f610b0d004adbdf63010dc4a73cb8c","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-04-29 14:39:09.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"09827c1cfdd543252a824e9fafe22d526724099f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-29 14:42:27.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/8/check-not-vote/ext-cinder-zuul/e9cb8b0 : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":8},{"id":"9049d0e9dedc1dcba82f777f7b90f010b0985cd9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-29 14:44:23.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/8/check/fujitsu-eternus-dx-iscsi/6dc45f0 : FAILURE in 35m 54s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/8/check/fujitsu-eternus-dx-fc/4f757b1 : FAILURE in 35m 26s","accounts_in_message":[],"_revision_number":8},{"id":"2f087d9508d09495d32cca05305ece98aa55b40d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-29 15:00:09.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"8eb015e4fcc8ac1df2c39003ce34aebb438d0086","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-29 15:08:28.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/49/713649/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/066eb49/066eb49.html : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":8},{"id":"541ec8421aa1f735d9299227a57e24db145b5995","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-29 15:18:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/8/iscsi/ : SUCCESS in 1h 10m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/8/smb/ : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":8},{"id":"4266af64a208e30e6f2f0e9ec7e8ab57943acdb4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-29 15:27:01.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b667a51 : FAILURE in 1h 18m 37s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/81f39bf : SUCCESS in 1h 16m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"473cce8d8ddb5f1afdb04a30ed9cdf9dc35f6eaa","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-29 15:32:14.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/49/713649/8/check/nimble-iscsi-driver-dsvm-volume/ce12006 : SUCCESS in 1h 23m 53s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/49/713649/8/check/nimble-fc-driver-dsvm-volume/551d52c : SUCCESS in 1h 16m 54s","accounts_in_message":[],"_revision_number":8},{"id":"bad78b86395a68b670daecbc45dd752b39f17efd","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-29 15:40:12.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"57de03fdf559b606b66e19d262f5816cde0afb15","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-29 15:45:25.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/8/check/DellEMC_XtremIO_FC/b9c2851/DellEMC_XtremIO_FC/1206 : FAILURE in 1h 32m 19s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/8/check/DellEMC_XtremIO_iSCSI/4dd0f23/DellEMC_XtremIO_iSCSI/1096 : SUCCESS in 1h 37m 02s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"d941004e836b4d4a853aead0e83fd0eca09bf8e0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-29 15:47:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7b1c384ca2b6f2837d7170dc6dd84877fe1e3e00","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-04-29 15:50:24.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"6cf258645e287164f29936bd78635c8fb7214abc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-29 16:08:18.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/9/iscsi/ : FAILURE in 20m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/9/smb/ : FAILURE in 20m 40s","accounts_in_message":[],"_revision_number":9},{"id":"e986ce336b5a21369794162a933030a60a77f357","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-29 16:39:59.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-8/ns5_nfs : SUCCESS in 2h 7m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-8/ns5_iscsi : SUCCESS in 2h 31m 30s","accounts_in_message":[],"_revision_number":8},{"id":"d37abb2abbdaa52b3240b6e31f301640b9214077","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-29 16:42:12.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/49/713649/9/check/fujitsu-eternus-dx-iscsi/d14ba7c : FAILURE in 35m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/49/713649/9/check/fujitsu-eternus-dx-fc/933b147 : FAILURE in 36m 38s","accounts_in_message":[],"_revision_number":9},{"id":"704e1404a3a1b93c4a1ce454549aacdab064e7f0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-29 16:44:57.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/49/713649/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/6f77d36/6f77d36.html : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":9},{"id":"3d845dfdb99109479e008c43501b1ca3f4b895b9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-29 16:53:35.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/9/check-not-vote/ext-cinder-zuul/97e34bd : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":9},{"id":"5c6864a5db8ec80c3d5c1a08ffd856eb0c5078d3","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-29 16:56:36.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"111829a938ec4a9c0ec778915035b849d4d59d67","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-29 17:55:32.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/9/check/tempest-dsvm-full-bionic/046fab6/ : SUCCESS in 1h 52m 39s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/9/check/devstack-plugin-ceph-dsvm-bionic/2a7fa92/ : SUCCESS in 1h 47m 15s","accounts_in_message":[],"_revision_number":9},{"id":"59181fd85be09553cff0d09e4d3c2108e9aa35c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 18:47:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98ea0c74f82e496bb40c0259011a29d1 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b88709952ed4b8fbdf74acc6dcf4690 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24238b1fb38c41d3b0e2011c5a5ff56c : SUCCESS in 9m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/980d775c1748444eba03a71cf90a5f0c : SUCCESS in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80addbecabe545669ee01a5794e71fcf : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82dbb5c53994eee9f915b9dd4c1ed6e : SUCCESS in 16m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/887f49efdf754a3cba9cdced1ab95313 : SUCCESS in 1h 07m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/909bbe8b0db34c1bb1609e54d7e91f0b : SUCCESS in 1h 12m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/448142959e6b4727ba1531737e1ee9ff : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f95eabf90f8e4c13ba3045660f92b99a : SUCCESS in 7m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e92c7e10d5254402bcbce866585060d7 : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be4d183f239a464890d1edfea73d38ac : SUCCESS in 51m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14a4019e8f9f4201b53e04214461134d : FAILURE in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a495e92659943a58e7370bbac8ba7b7 : SUCCESS in 1h 18m 46s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8f33aa867aac47c889bc40b51e58859e : TIMED_OUT in 2h 06m 50s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/03c15169d10748f192115126d947b7d5 : SUCCESS in 1h 04m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2657a0c56cd64bf19bd9a9cd85fed3ea : SUCCESS in 1h 27m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f875ada6521d41a0b4ee64512cca7ea6 : SUCCESS in 1h 18m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16361e70dc1142a8ab65ce81c7833aa5 : SUCCESS in 2h 14m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e10d6e50f4bf451f9609dc3589ff27e4 : SUCCESS in 1h 10m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa363591f577483e8435e6411e67cdcc : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":9},{"id":"34f1051e81a503b230dfbf4f458dcaf4f7940912","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-29 18:56:51.000000000","message":"Patch Set 9:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-9/ns5_nfs : SUCCESS in 2h 23m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-9/ns5_iscsi : SUCCESS in 2h 26m 39s","accounts_in_message":[],"_revision_number":9},{"id":"cc406752000b5e131b176a7604615b26e061fbf4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-29 19:34:59.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 32m 17s","accounts_in_message":[],"_revision_number":9},{"id":"0221c5610dbf0aa470419e8eba55e9432f6255a5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-29 20:08:53.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/9/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3926fff : FAILURE in 1h 52m 30s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/9/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/88f2f5f : SUCCESS in 1h 07m 11s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"be8abb9edd49863f6a55ca10a4182266c49bae12","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-29 20:44:25.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/9/check/DellEMC_XtremIO_FC/b852408/DellEMC_XtremIO_FC/1209 : SUCCESS in 1h 32m 16s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/9/check/DellEMC_XtremIO_iSCSI/1e33abb/DellEMC_XtremIO_iSCSI/1099 : SUCCESS in 1h 39m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"ebec2491f06af45c84708366a4f808f1a89897c2","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-29 20:51:53.000000000","message":"Patch Set 9:\n\n(3 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"ca03701f3df8b104a2dd17610ab231176b2b9844","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-29 21:43:03.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"d52f592194cb892914af92685ab774496aed609b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-29 22:16:03.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/49/713649/9/check/DellEMC_Unity_FC/3bc56b2/DellEMC_Unity_FC/1177 : FAILURE in 1h 35m 11s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/49/713649/9/check/DellEMC_Unity_iSCSI/608140e/DellEMC_Unity_iSCSI/1053 : SUCCESS in 1h 58m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"1297ed23cdffa77a68d425b780ba6d74192c0917","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-30 00:02:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/49/713649/9/check/nimble-iscsi-driver-dsvm-volume/7e76c38 : SUCCESS in 1h 24m 07s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/49/713649/9/check/nimble-fc-driver-dsvm-volume/faacdb0 : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":9},{"id":"8dfe6d2d8813ee516276db7dbe59c6904827734d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-30 00:03:06.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"266708acddbcab34ef3830387f8713f1360b40a4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-30 03:45:19.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"96fb8f78afa3f9762bda779c64e96230cb867bbb","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-30 06:52:24.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"d3a7ceda2a84332a40c6815902ac8c1e6ccfd17c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-30 07:28:03.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"70e5b599c91790357bdda678be5f22c224cb3cd9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-30 07:30:45.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/9/check-not-vote/ext-cinder-zuul/50b638e : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":9},{"id":"9a264cb2220a9b33813d0421fcc0be2effce0006","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-30 07:43:36.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"4f37ac68b2a899c03c1437af31802095b6288f19","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-30 07:54:26.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/9/iscsi/ : SUCCESS in 1h 01m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/9/smb/ : FAILURE in 44m 15s","accounts_in_message":[],"_revision_number":9},{"id":"07436d9eecc54a0b92a32df692dab8255ece07c7","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-30 08:01:15.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/9/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3961e49 : SUCCESS in 58m 30s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/9/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/5b1dada : SUCCESS in 1h 08m 32s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"ba4cb3680f20d5bdd9336307047c1e1bf6b6ca09","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-30 08:14:13.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 1h 06m 31s","accounts_in_message":[],"_revision_number":9},{"id":"87f292a2bdb1b26dfeb976fbc276c5ae9909da15","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-30 09:16:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/882/index.html : SUCCESS in 1h 16m 31s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1031/index.html : SUCCESS in 1h 42m 25s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1070/index.html : SUCCESS in 1h 26m 26s","accounts_in_message":[],"_revision_number":9},{"id":"1b6ea5958e3c0bfa3b6f3eb5dddf221685c2b4ab","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-30 09:23:12.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"66a79ee1763447a6e4c4e26eb551ce6732995d50","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-30 09:36:51.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/9/check/tempest-dsvm-full-bionic/690a44e/ : SUCCESS in 2h 02m 13s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/9/check/devstack-plugin-ceph-dsvm-bionic/453403c/ : SUCCESS in 1h 45m 52s","accounts_in_message":[],"_revision_number":9},{"id":"24870a1faee65be641947b1297a86c52494e56b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 09:38:48.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c81658a051e474296abb648f5238c03 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f70a038328747198d9c853bb916e12a : SUCCESS in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/625eacd300e64f4ca1cbddf96331ea3b : SUCCESS in 11m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51e008378043482eb78065a11ce9d78f : SUCCESS in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/891799cc291243359bcfc49f242a627a : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab11d1ace1446a4aa504e2f8008b345 : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/c01201f7247a48188f17bc41df081ac4 : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8fb77d4cf7e44727af5e70cc3c8e5bc6 : SUCCESS in 1h 11m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36daf2a77c2048c296ce5033e21ebbd5 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee122fecac1c40ed9bf2264fcf1d179c : SUCCESS in 6m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4b2e243c0dc344dfa4c3f8ff997bf10c : SUCCESS in 10m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8d01195661a24230ba138829b694c311 : SUCCESS in 41m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2104ada186e48c797ae7a95a4668165 : FAILURE in 9m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1d0a76ffffa24d1f9bdc04152198f7b1 : SUCCESS in 1h 13m 26s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b565c85330e54e3fac452371cd73140f : TIMED_OUT in 2h 08m 12s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e8ea0b04d7c34328862585df38bbe7fb : SUCCESS in 1h 12m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/12f4c5f9642b4c6b8b903fe38648242a : SUCCESS in 1h 25m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/901a1d98e45343c68dce4d6cd343c885 : SUCCESS in 1h 22m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/00e7be8c491442208d3a763fee7c1904 : FAILURE in 2h 42m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bf9c48f6e98d489f9b242c46d40032a6 : SUCCESS in 1h 02m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ae14d47347d40cca0f68d9289404e82 : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":9},{"id":"f9330df18f0631e6a580dbebb61b1667d96ee468","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-30 09:43:45.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"32610908d8665bfd8908eb4853cd73036d94b036","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-30 10:20:45.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/9/check-not-vote/ext-cinder-zuul/5949013 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":9},{"id":"b2e475d08405d73bb6aabae4afa44bbcd063395c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-30 10:34:37.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"51b835ba7b9b641253f07660887cbf18c9deab8b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-30 10:44:42.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/9/iscsi/ : SUCCESS in 1h 00m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/9/smb/ : FAILURE in 17m 54s","accounts_in_message":[],"_revision_number":9},{"id":"581fd9e1fad5b7da0cd39afc40ec0c2b3de60d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 11:53:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11f16ae20dd3411eab5508cf9041a539 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434825f2fd1e452d98f28d2ceaa38a13 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1d588dbc57445b6b66e9bdfcaf00f11 : SUCCESS in 10m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/807d86c25aef498aba51451aa3c8fd62 : SUCCESS in 11m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/513feb21dae94f4ea4229f7f275c244d : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca890b16c0c845dab39cb5d3b6b66b52 : SUCCESS in 14m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/04b2988ac738426a9292277acbe948f3 : SUCCESS in 1h 02m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7deadcf4c2834f3685e21c8bf9c3d2cc : SUCCESS in 1h 13m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93343709faf84f4b9924ce6940591158 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/553adbcd8b6c4b28b6393ae2e789571f : SUCCESS in 5m 59s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/05e3ecf1bca04b85b8ab2fbe9e6d8913 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d37ff76cd8914fcbbdc4d8718b21180e : SUCCESS in 52m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/388d346ca1d249bfb0b986192687e36d : FAILURE in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f85a8ad98da041faa7959732161408c7 : SUCCESS in 1h 11m 23s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5126c54471904917a5e7d84bc4bf7296 : SUCCESS in 1h 42m 04s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1181b16fdb8042fbab984bc64a9b60fa : SUCCESS in 1h 06m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8cb26a11b7ec4c9cbdad0b54dd342989 : SUCCESS in 1h 11m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/675a139468ae4b7cabf8864f6161b657 : SUCCESS in 1h 12m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fef8cedf9f45486887826be0444177cb : SUCCESS in 2h 07m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/60f92391d2f24555b011cde309c13ba2 : SUCCESS in 1h 00m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e60ee07fac9e4918b81895cf193e6f0f : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":9},{"id":"94e7c75c57ded906892350330cbbc64c35c197d3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-30 12:28:30.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/9/check/tempest-dsvm-full-bionic/7d6b0a2/ : SUCCESS in 1h 53m 50s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/9/check/devstack-plugin-ceph-dsvm-bionic/569c4a3/ : SUCCESS in 2h 03m 24s","accounts_in_message":[],"_revision_number":9},{"id":"a23d2b3cdad133a05ed49637fa6c2e6e179c0b18","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-30 12:41:19.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/143/index.html : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"82e34f5c436d80a7c528ba6a7c491e4ac6b9222d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-30 12:57:44.000000000","message":"Patch Set 9:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-49-713649-9 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":9},{"id":"6605824b07bee715ade3b615b9a79aa6a48ad695","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-04-30 13:06:34.000000000","message":"Patch Set 9:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/713649-9/Lenovo-iSCSI : FAILURE in 23:20.76\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/713649-9/Lenovo-FC : FAILURE in 19:10.48","accounts_in_message":[],"_revision_number":9},{"id":"cc03293c0405898b7fb7f56af855be995c6f110c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-04-30 13:53:55.000000000","message":"Patch Set 9:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/713649-9/Seagate-iSCSI : SUCCESS in 47:16.61\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/713649-9/Seagate-FC : SUCCESS in 41:31.67","accounts_in_message":[],"_revision_number":9},{"id":"e612d0bd28bf2fd60546bc7967bd37454acbf311","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-30 14:21:08.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/49/713649/9/check/DellEMC_VxFlexOS/4e7f017/DellEMC_VxFlexOS/725 : SUCCESS in 2h 20m 04s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":9},{"id":"0064cf768bd2b56b0d4f87b05a3eee56b91698bc","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-30 14:37:31.000000000","message":"Patch Set 9:\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":9},{"id":"29897ebcf91bfb0963457d7cd601636efd442c6b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-30 14:37:46.000000000","message":"Patch Set 9:\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":9},{"id":"399da9f9adc2a2e642842ccd275a642056f047ce","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-30 14:41:34.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/9/check/DellEMC_XtremIO_FC/cf794af/DellEMC_XtremIO_FC/1217 : SUCCESS in 1h 31m 20s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/9/check/DellEMC_XtremIO_iSCSI/f9190c3/DellEMC_XtremIO_iSCSI/1107 : SUCCESS in 1h 38m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"4977765cde5df255348784a2d556256fe7e397bf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-30 15:18:47.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/49/713649/9/check/DellEMC_Unity_FC/a27d0f3/DellEMC_Unity_FC/1184 : SUCCESS in 1h 26m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/49/713649/9/check/DellEMC_Unity_iSCSI/c4500bf/DellEMC_Unity_iSCSI/1060 : SUCCESS in 1h 52m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"26e6f3a371ca86b09294391879286530ba5c7878","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-04-30 15:30:37.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/49/713649/9/check/DellEMC_SC_FC/77c29b4/DellEMC_SC_FC/717 : SUCCESS in 1h 09m 35s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/49/713649/9/check/DellEMC_SC_iSCSI/8e4b85e/DellEMC_SC_iSCSI/715 : FAILURE in 2h 50m 45s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"ff1a78876422ab0d51595581aaa327c840368198","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-30 15:43:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/9/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ec10f22 : SUCCESS in 1h 00m 57s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/9/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/89371f7 : SUCCESS in 1h 08m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"e825bfeea024f99d4ba5983e1b77da13d5fe412d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-30 15:57:45.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 1h 03m 52s","accounts_in_message":[],"_revision_number":9},{"id":"eabf9148b4584a777c6e4babc1408f3aea3ec4eb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-30 21:00:39.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/49/713649/9/upstream-check/cinder-cDOT-iSCSI/0ed3afb/ : FAILURE in 3h 49m 14s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/49/713649/9/upstream-check/cinder-cDOT-iSCSI-IPv6/0b79d98/ : SUCCESS in 3h 10m 59s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/49/713649/9/upstream-check/cinder-cDOT-NFS/6dfcaf2/ : FAILURE in 3h 49m 12s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/49/713649/9/upstream-check/cinder-cDOT-NFS-IPv6/b35ad77/ : FAILURE in 3h 54m 52s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/49/713649/9/upstream-check/cinder-cDOT-FCP/bfc953c/ : FAILURE in 3h 19m 50s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"c2e3f5bec36fe0d07d4493e1cecc3b6cb023d5c4","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-05-01 03:12:58.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/890/index.html : SUCCESS in 1h 15m 39s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1039/index.html : SUCCESS in 1h 38m 22s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1078/index.html : FAILURE in 1h 44m 55s","accounts_in_message":[],"_revision_number":9},{"id":"c7f10f6a140b9ee956ed343c2d73dcaedc8019af","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-01 08:21:31.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"a4f43f8761d530d8204441658e3c69b6cce817c4","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-05-02 17:37:03.000000000","message":"Patch Set 9:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/49/713649/9/check/kaminario-dsvm-tempest-full-iscsi/f1f1880 : FAILURE in 2h 16m 04s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/49/713649/9/check/kaminario-dsvm-tempest-full-FC/9f80872 : FAILURE in 2h 01m 22s","accounts_in_message":[],"_revision_number":9},{"id":"d8893c16120f68ed76c305b53ab5157bcb576b39","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-05-02 22:15:29.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/49/713649/9/check/cinder-storpool-tempest/57d0573/ : SUCCESS in 2h 15m 44s","accounts_in_message":[],"_revision_number":9},{"id":"1f8ebe06316a032e4952cb70ac59f2c52e82a991","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-05-02 22:58:58.000000000","message":"Patch Set 8:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/49/713649/8/check/kaminario-dsvm-tempest-full-iscsi/eac5733 : FAILURE in 2h 19m 28s\n- kaminario-dsvm-tempest-full-FC http://52.3.254.104/49/713649/8/check/kaminario-dsvm-tempest-full-FC/4b27b4b : FAILURE in 2h 01m 58s","accounts_in_message":[],"_revision_number":8},{"id":"c59410ef04571dffcc139c475a3f09d79383dd45","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-05-09 05:23:07.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/9/20200509114945/NEC-FC : SUCCESS in 2h 25m 48s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/9/20200509114945/NEC-ISCSI : SUCCESS in 2h 33m 19s","accounts_in_message":[],"_revision_number":9},{"id":"c8e1e15a0336b17b80c9c5c1a56e31d87319f0d1","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-22 07:47:46.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"5ae8fe6668f75d88fbd3b722cedd992ad0c4942c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-22 07:49:50.000000000","message":"Topic changed from bug/1867906 to a-a-support","accounts_in_message":[],"_revision_number":10},{"id":"043d522da25782ecee1e753556c5364829b472c0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-22 08:09:42.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/10/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/99aa4d6 : FAILURE in 21m 32s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/10/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/1d889d2 : FAILURE in 20m 37s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"a6cbec9158946c3f80f8661d078d65392e607509","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 08:26:19.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/10/check-not-vote/ext-cinder-zuul/a4a6f8d : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":10},{"id":"c38ee077f9963795d41e32c174d61320c3e7ef22","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-22 08:38:38.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"ac7771b79968613cf11cd212efb72a61561ed335","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-22 08:50:50.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/10/iscsi/ : FAILURE in 1h 02m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/10/smb/ : FAILURE in 56m 10s","accounts_in_message":[],"_revision_number":10},{"id":"eda47e30c65e4c41c48ba66af8190c74b5834a90","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-22 08:58:25.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/49/713649/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/69a809f/69a809f.html : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":10},{"id":"e17d356a6e25b09bef1ec23aaa4bfb050c482550","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-22 09:04:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/49/713649/10/check/dsvm-tempest-cisco-zonemanager-job/a96c366/ : FAILURE in 33m 36s","accounts_in_message":[],"_revision_number":10},{"id":"8ad18e259f7903b7cb6c9106c411f29f219cfcf7","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-22 09:05:11.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1734 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"680d5720f174b70f5b093fc335e2953b1f2a9398","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 09:18:38.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"c16872f465436368af458bb804105771724256d2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-22 10:18:26.000000000","message":"Patch Set 10:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-49-713649-10/ns5_nfs : SUCCESS in 2h 10m 19s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-49-713649-10/ns5_iscsi : SUCCESS in 2h 30m 13s","accounts_in_message":[],"_revision_number":10},{"id":"6c57815699c001a5c5892e49bbcc77b01f200791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 11:13:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92aa4a452f784b4a86d27ae6b21cfb07 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d9a566f605495d82837d5e19ce5b39 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5318811af3e94d67ab57ee07cbe72e1f : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81a72befef134cab91ce4096442267d6 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677f4127346a409e9051bbae25b9206e : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/735b45d7251741fe93a504dd09c781f3 : SUCCESS in 1h 05m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11246874c90044dba647a8fb2cae054c : SUCCESS in 1h 17m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d5a749afb55413f9421f57aa010b8eb : SUCCESS in 12m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4798e6bd8a8e4b7a8f3e9c53c7648bf3 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccf73e84e0b841bd9a8ed0c73ef4ff49 : SUCCESS in 6m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c6d14a9118b444088dd02edd920aecc : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e672096899e4469585a119fb7caef0e8 : SUCCESS in 44m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf1d2b1406704dd79a719f16df9566dd : SUCCESS in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e1c2d398e1bd4e23a076d76a833173b4 : SUCCESS in 1h 22m 53s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e805bd33c5b443a2a160b0726f76f2bf : SUCCESS in 1h 22m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fea4010ed1464dac96916ed5194fe69e : SUCCESS in 1h 07m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a1d84d528f27407fa331c4941364a630 : SUCCESS in 1h 23m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a6f00c6c2ca644bf80169f3353040d3f : SUCCESS in 1h 17m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45d3f02febc74789aa844c067d7fd3d5 : FAILURE in 2h 45m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c34e8522a68042418989dbbb04a25dd9 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":10},{"id":"b164b7e790fc366c579757c6fa9cc099b625d8ab","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 11:46:12.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"a188e23c8303525b13e70ede925e22d43f742f6a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-22 12:24:03.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/10/check/DellEMC_XtremIO_FC/b812584/DellEMC_XtremIO_FC/465 : SUCCESS in 1h 31m 37s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/10/check/DellEMC_XtremIO_iSCSI/0437c29/DellEMC_XtremIO_iSCSI/437 : SUCCESS in 1h 37m 52s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"f175cf5f64a0e9c488e83bbd13f7209925342f49","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-22 12:29:03.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/8d67562831606d048f6f381a5b94814216664fbe/index.html : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"0ecae63c200ffeb675c3ca291836012a9a76bd36","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-22 14:35:57.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"e922261d7ab6bd0cd3f3ca1c6c21dc0bcd3edb7a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-22 15:25:10.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/49/713649/10/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/843d879 : FAILURE in 22m 31s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/49/713649/10/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/086af26 : FAILURE in 21m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"a3d6d0880c348fe2c0117c1f9a25c30103bb9d72","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 15:25:17.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"0b762ca496893731c77298096e6c64e7f483e3c7","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-22 15:27:56.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"5740200b2d85b843942fd6c5821380a1b6f063d4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-22 15:45:05.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/713649/10/iscsi/ : FAILURE in 1h 01m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/713649/10/smb/ : FAILURE in 1h 08m 43s","accounts_in_message":[],"_revision_number":10},{"id":"2d0bc2d197e884267e72c4da66885cb2a21004ee","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-22 15:48:27.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/49/713649/10/check/dsvm-tempest-cisco-zonemanager-job/d32d09a/ : FAILURE in 32m 41s","accounts_in_message":[],"_revision_number":10},{"id":"2f35e575d216754cd912e7783ccf8beca9120184","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 17:15:35.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/49/713649/10/check-not-vote/ext-cinder-zuul/2fbef19 : SUCCESS in 37m 10s","accounts_in_message":[],"_revision_number":10},{"id":"626bc8d85d1268c98895ca17995e2ba8f3bc2993","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-22 17:27:02.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/49/713649/10/check/dsvm-tempest-zadara-driver/3b41ae9 : FAILURE in 1h 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5728d4238f4e56bfadc82574f8dbbff5e1166ce1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 18:34:08.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"1405913d7a8bda5c3751639422563f9c0a75221c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:03:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8863da9c1de4dc0a76dcc8bba86ae30 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d4c39297af448589be89d5053558ca4 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8df3d937d84e4a5b86c1bb433a2f835a : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb3601a98cc14cdcbcdcac683e569b4f : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a67ce074e844dcb5957ef345abda3d : SUCCESS in 17m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/608f71ddd9e44821bf66413fe014e20e : SUCCESS in 1h 03m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0971fa405f77467ab95d59b15a872c40 : SUCCESS in 54m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0eda033b742b44b7b94e61e93e0b374c : SUCCESS in 13m 48s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8fa43fb41260449ebfd3567df7d37e53 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7333301f29d14e2fa555e9d724666d95 : SUCCESS in 6m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/732644f769074c8f8933733c92d7a64a : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6606cb75403e4968a9b36c2c3ac9cd21 : SUCCESS in 46m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8cfd8c1e0014bc2aaf1fd74f07109c4 : SUCCESS in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/27eb30f3d21b4678afdb026ac0a95911 : SUCCESS in 1h 19m 27s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c80f747d807a4bca8fe00ae8f5afb08c : SUCCESS in 1h 35m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3df5ef6f48104f6cbc20dba26d6122f1 : SUCCESS in 1h 05m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a12dfea573b24d81b2d54b11d651167d : SUCCESS in 1h 11m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0584344950c54e0a93cf8c94a8a728dd : SUCCESS in 1h 20m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8e8d93b69f4540428472658ff8608992 : SUCCESS in 2h 06m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a861792e9534163a0854d6a81fbaea8 : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":10},{"id":"71d38eecd957db1cd94fbccfcb15d9d1d538526c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-22 19:32:01.000000000","message":"Patch Set 10:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-49-713649-10 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":10},{"id":"2c175085363a2404a952b1574e3b08b7ec36a439","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-22 20:00:14.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/49/713649/10/check/synology-dsm-cinder-iscsi/cc013f7 : FAILURE in 27m 46s","accounts_in_message":[],"_revision_number":10},{"id":"8b6f90df98d04c2c5372c5753df2c547d7a6d4d1","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-22 20:30:30.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/49/713649/10/check/nimble-iscsi-driver-dsvm-volume/9e8db35 : FAILURE in 50m 43s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/49/713649/10/check/nimble-fc-driver-dsvm-volume/36b4842 : SUCCESS in 1h 57m 30s","accounts_in_message":[],"_revision_number":10},{"id":"bbf8e3220f7ab3a4b4f4d06af7cf87eab6259a22","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-22 20:39:43.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/49/713649/10/upstream-solidfire-check/cinder-solidfire-iSCSI/b70bebb/ : SUCCESS in 1h 34m 48s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"fc168af1545781229bfd402cdbec9ffe7a895454","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 22:15:50.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"b159f06fe37929644cae807601bb4004005bed91","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-23 01:16:11.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/49/713649/10/check/DellEMC_PowerFlex/bd12f5e/DellEMC_PowerFlex/295 : SUCCESS in 1h 56m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"41f2ac0ba437c9b02b1aa87cedb8b8ddd811e42e","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-07-23 01:46:42.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"40fa9e0d54e4f895ce4ce0554f01ca38f4d43468","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-23 02:55:59.000000000","message":"Patch Set 10:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/713649-10/HPMSA-iSCSI : SUCCESS in 43:50.10\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/713649-10/HPMSA-FC : SUCCESS in 44:11.85","accounts_in_message":[],"_revision_number":10},{"id":"03f80c3543f0ece2e06436c5158a741237171d6b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 02:56:05.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"9d903c212c58e8399d6c3c49a4eef581a138291a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-23 03:03:20.000000000","message":"Patch Set 10:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/713649-10/Seagate-iSCSI : SUCCESS in 43:43.73\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/713649-10/Seagate-FC : SUCCESS in 42:28.07","accounts_in_message":[],"_revision_number":10},{"id":"d3f9775721943232411ff24cb59f50d6db17e5b5","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-23 03:40:58.000000000","message":"Patch Set 10:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/713649-10/Lenovo-iSCSI : SUCCESS in 46:29.69\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/713649-10/Lenovo-FC : SUCCESS in 44:34.53","accounts_in_message":[],"_revision_number":10},{"id":"a9ba89e58511cd9e8f6e46e807cccb542c70377c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-07-23 03:53:32.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/49/713649/10/check/DellEMC_SC_FC/97e8e27/DellEMC_SC_FC/215 : SUCCESS in 1h 19m 15s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/49/713649/10/check/DellEMC_SC_iSCSI/35c2fdc/DellEMC_SC_iSCSI/188 : FAILURE in 1h 35m 13s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"ddea48969584f153060a035d381c1f7f56790f8c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 04:26:57.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"649ab5a9ec3274a82b29567fa66266171be53a65","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-23 10:17:01.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/10/check/tempest-dsvm-full-bionic/3e67ac5/ : SUCCESS in 1h 43m 47s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/49/713649/10/check/devstack-plugin-ceph-dsvm-bionic/2b45112/ : SUCCESS in 1h 56m 20s","accounts_in_message":[],"_revision_number":10},{"id":"6693078d79553d54be3bc323d01bb9435b1908d4","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-23 11:13:51.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/10/20200723173931/NEC-ISCSI : SUCCESS in 2h 23m 35s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/49/713649/10/20200723173931/NEC-FC : SUCCESS in 2h 34m 17s","accounts_in_message":[],"_revision_number":10},{"id":"2342361c4586599ab353df4091e15b98b6d7d610","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 11:55:23.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"05f54871fd48e1f3e6eeadff113d93770068769d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-23 12:08:54.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/49/713649/10/check/DellEMC_VNX_iSCSI/0ceb070/DellEMC_VNX_iSCSI/174 : SUCCESS in 2h 46m 16s\n- DellEMC_VNX_FC http://publiclogs.emc.com/49/713649/10/check/DellEMC_VNX_FC/a02953f/DellEMC_VNX_FC/168 : SUCCESS in 1h 07m 55s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"23082231bb3e746d30961f9ba9ef8814356ad3a0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 13:14:25.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"1600b990d7e5e738bae888733cbadadff6c9a39d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-23 13:54:53.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/49/713649/10/check/DellEMC_XtremIO_FC/d843061/DellEMC_XtremIO_FC/478 : SUCCESS in 1h 34m 10s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/49/713649/10/check/DellEMC_XtremIO_iSCSI/ef38027/DellEMC_XtremIO_iSCSI/450 : SUCCESS in 1h 33m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"a7e7edd40ebe84526710f3adf803ee835218de75","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 15:05:02.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/49/713649/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"c3e929fc94f129a817c0d39fdc10e93543b5dbf2","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-07-28 16:47:52.000000000","message":"Patch Set 10:\n\ne0ne, if you have the chance please add the missing release note","accounts_in_message":[],"_revision_number":10},{"id":"74cf6ff5fcefaa3efe6403fb0790515a63f04e08","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-08-04 13:45:32.000000000","message":"Patch Set 10:\n\nGorka, I\u0027m going to add release note to the next patch https://review.opendev.org/#/c/724695/2 to catch all related changes","accounts_in_message":[],"_revision_number":10},{"id":"9e1a29ac373915c232b01f02c145cf337a65300e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-07 02:16:58.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"af004374e49b5d98d390f4361cbbff3d6c590f46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 02:45:24.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ee26bc87b7b52eba723967028a2110a3d23312e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 06:26:42.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17321df443bf4431aced411ba6c71d2b : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a08841c976274984b03773223b7ad4fd : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bea7cf5e387c4d359a4f5e640dce0e8d : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7899f271d1984ff5adfb0f1b19513450 : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35af88cb219a430698e310f7d7e3220d : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/bc448ca79fa24b7dbf3e572a4164464f : SUCCESS in 52m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2107965c25e49dda74881b4780aa3e3 : SUCCESS in 59m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1f3639aef1154427bae76085543fd519 : SUCCESS in 1h 02m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5f0777965bab49cd88eba4117aa20216 : SUCCESS in 1h 06m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e17fa1c0f7a1410f9eeafe7dabf5d175 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":10},{"id":"20550126a84f82a575d245318ff39526d4ec679f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 06:26:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8d67562831606d048f6f381a5b94814216664fbe","revisions":{"7dac67f9357dea39f03bbd84703113ea85e86a65":{"kind":"REWORK","_number":1,"created":"2020-03-18 13:11:23.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/1"}}},"commit":{"parents":[{"commit":"3f3b7189dba60a2b1fd336241ac8839cd09098ea","subject":"Merge \"Tests: Remove Python 2 compat file_spec code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f3b7189dba60a2b1fd336241ac8839cd09098ea"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"subject":"WIP.  FilterScheduler uses backend.host instead backend.id now","message":"WIP.  FilterScheduler uses backend.host instead backend.id now\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7dac67f9357dea39f03bbd84703113ea85e86a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7dac67f9357dea39f03bbd84703113ea85e86a65"}]},"branch":"refs/heads/master"},"250a402f061eac64bf1d8f20a4cd5620f7654bba":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-18 18:41:53.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/2"}}},"commit":{"parents":[{"commit":"3f3b7189dba60a2b1fd336241ac8839cd09098ea","subject":"Merge \"Tests: Remove Python 2 compat file_spec code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f3b7189dba60a2b1fd336241ac8839cd09098ea"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 18:41:49.000000000","tz":0},"subject":"FilterScheduler uses backend.host instead backend.id now","message":"FilterScheduler uses backend.host instead backend.id now\n\nBoth volumes and volume groups has backend hostname instead of backend\nID in the database. These values are equal in a usual deployment but in\nan Active/Active mode backend ID contains cluster name. That\u0027s why we\nneed to use hostname explicitly in a scheduler.\n\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/250a402f061eac64bf1d8f20a4cd5620f7654bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/250a402f061eac64bf1d8f20a4cd5620f7654bba"}]},"branch":"refs/heads/master"},"94850982d5ba9d2c87aee49fcf4c37a928426406":{"kind":"REWORK","_number":3,"created":"2020-03-19 11:32:45.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/3"}}},"commit":{"parents":[{"commit":"3f3b7189dba60a2b1fd336241ac8839cd09098ea","subject":"Merge \"Tests: Remove Python 2 compat file_spec code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f3b7189dba60a2b1fd336241ac8839cd09098ea"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-19 11:32:13.000000000","tz":120},"subject":"WIP. FilterScheduler uses backend.host instead backend.id now","message":"WIP. FilterScheduler uses backend.host instead backend.id now\n\nTemporary commit to test on CI\nTODO(e0ne): update commit message.\n\nBoth volumes and volume groups has backend hostname instead of backend\nID in the database. These values are equal in a usual deployment but in\nan Active/Active mode backend ID contains cluster name. That\u0027s why we\nneed to use hostname explicitly in a scheduler.\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94850982d5ba9d2c87aee49fcf4c37a928426406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94850982d5ba9d2c87aee49fcf4c37a928426406"}]},"branch":"refs/heads/master"},"1c83de50b52162170281bbc077b1d6cbc2f61adc":{"kind":"REWORK","_number":4,"created":"2020-03-19 15:30:18.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/4"}}},"commit":{"parents":[{"commit":"3f3b7189dba60a2b1fd336241ac8839cd09098ea","subject":"Merge \"Tests: Remove Python 2 compat file_spec code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f3b7189dba60a2b1fd336241ac8839cd09098ea"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-19 15:25:20.000000000","tz":120},"subject":"Set cluster name for volume groups","message":"Set cluster name for volume groups\n\nIn Active/Active HA mode we need to store not only worker hostname but\ncluster name too. This patch saves cluster name for groups created from other\ngroups or snaphshots.\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c83de50b52162170281bbc077b1d6cbc2f61adc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c83de50b52162170281bbc077b1d6cbc2f61adc"}]},"branch":"refs/heads/master"},"06fda3d644f11f4be8e3f240683d6492733722eb":{"kind":"REWORK","_number":5,"created":"2020-03-25 13:05:43.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/5"}}},"commit":{"parents":[{"commit":"3f3b7189dba60a2b1fd336241ac8839cd09098ea","subject":"Merge \"Tests: Remove Python 2 compat file_spec code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f3b7189dba60a2b1fd336241ac8839cd09098ea"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-25 13:05:26.000000000","tz":120},"subject":"Set cluster name for volume groups","message":"Set cluster name for volume groups\n\nIn Active/Active HA mode we need to store not only worker hostname but\ncluster name too. This patch saves cluster name for groups created from other\ngroups or snaphshots.\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06fda3d644f11f4be8e3f240683d6492733722eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06fda3d644f11f4be8e3f240683d6492733722eb"}]},"branch":"refs/heads/master"},"b547a8a203bf061368998f777abf2e239b6f1e2b":{"kind":"REWORK","_number":6,"created":"2020-04-24 15:53:28.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/6"}}},"commit":{"parents":[{"commit":"3f3b7189dba60a2b1fd336241ac8839cd09098ea","subject":"Merge \"Tests: Remove Python 2 compat file_spec code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f3b7189dba60a2b1fd336241ac8839cd09098ea"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-24 15:53:08.000000000","tz":180},"subject":"Set cluster name for volume groups","message":"Set cluster name for volume groups\n\nIn Active/Active HA mode we need to store not only worker hostname but\ncluster name too. This patch saves cluster name for groups created from other\ngroups or snaphshots.\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b547a8a203bf061368998f777abf2e239b6f1e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b547a8a203bf061368998f777abf2e239b6f1e2b"}]},"branch":"refs/heads/master"},"4e682333fd273850d2b4f3246e5ddee3dd99cda3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-04-24 21:42:28.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/7"}}},"commit":{"parents":[{"commit":"50aebee62cc79e3956afbe2ea9f3cc71046000be","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50aebee62cc79e3956afbe2ea9f3cc71046000be"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-24 21:42:28.000000000","tz":0},"subject":"Set cluster name for volume groups","message":"Set cluster name for volume groups\n\nIn Active/Active HA mode we need to store not only worker hostname but\ncluster name too. This patch saves cluster name for groups created from other\ngroups or snaphshots.\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e682333fd273850d2b4f3246e5ddee3dd99cda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e682333fd273850d2b4f3246e5ddee3dd99cda3"}]},"branch":"refs/heads/master"},"ed3216ab4c459e1ba2e33eb55fd02fa757a88d0c":{"kind":"REWORK","_number":8,"created":"2020-04-29 14:08:03.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/8"}}},"commit":{"parents":[{"commit":"50aebee62cc79e3956afbe2ea9f3cc71046000be","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50aebee62cc79e3956afbe2ea9f3cc71046000be"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-29 14:07:54.000000000","tz":0},"subject":"Set cluster name for volume groups","message":"Set cluster name for volume groups\n\nIn Active/Active HA mode we need to store not only worker hostname but\ncluster name too. This patch saves cluster name for groups created from other\ngroups or snaphshots.\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed3216ab4c459e1ba2e33eb55fd02fa757a88d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed3216ab4c459e1ba2e33eb55fd02fa757a88d0c"}]},"branch":"refs/heads/master"},"0a7c8de3294180406d70c2c3f18d5138737ea1a9":{"kind":"REWORK","_number":9,"created":"2020-04-29 15:47:03.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/9"}}},"commit":{"parents":[{"commit":"50aebee62cc79e3956afbe2ea9f3cc71046000be","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50aebee62cc79e3956afbe2ea9f3cc71046000be"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-29 15:46:51.000000000","tz":180},"subject":"Set cluster name for volume groups","message":"Set cluster name for volume groups\n\nIn Active/Active HA mode we need to store not only worker hostname but\ncluster name too. This patch saves cluster name for groups created from other\ngroups or snaphshots.\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a7c8de3294180406d70c2c3f18d5138737ea1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a7c8de3294180406d70c2c3f18d5138737ea1a9"}]},"branch":"refs/heads/master"},"8d67562831606d048f6f381a5b94814216664fbe":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-07-22 07:47:46.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/49/713649/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/713649/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/713649/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/713649/10"}}},"commit":{"parents":[{"commit":"8bf454b0b5c7b77bdde8fe8a98d5225957e9c912","subject":"Merge \"Don\u0027t show host_name to non-admins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bf454b0b5c7b77bdde8fe8a98d5225957e9c912"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-18 13:10:39.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-07-22 07:45:14.000000000","tz":180},"subject":"Set cluster name for volume groups","message":"Set cluster name for volume groups\n\nIn Active/Active HA mode we need to store not only worker hostname but\ncluster name too. This patch saves cluster name for groups created from other\ngroups or snaphshots.\n\nChange-Id: I6f160cc44350f0d378fdddb7943e8cdcc15be1b8\nCloses-Bug: #1867906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d67562831606d048f6f381a5b94814216664fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d67562831606d048f6f381a5b94814216664fbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
