)]}'
{"id":"openstack%2Fcinder~722081","triplet_id":"openstack%2Fcinder~master~I2dd06c43c39ddb576fea9e5cbccffc096e77073a","project":"openstack/cinder","branch":"master","topic":"bug/1860913","hashtags":[],"change_id":"I2dd06c43c39ddb576fea9e5cbccffc096e77073a","subject":"nfs: Mark the initialize_connection API as idempotent in connection_info","status":"ABANDONED","created":"2020-04-22 17:02:38.000000000","updated":"2021-07-05 17:19:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21d241605d1531ba17a6f3b7cb306edbd4faacd8","_number":722081,"virtual_id_number":722081,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-18 20:12:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-07-05 09:41:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 12:37:05.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"REVIEWER"},{"updated":"2020-05-06 11:03:50.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-06 11:05:38.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-06 11:18:45.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-05-06 13:54:27.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-05-06 14:02:47.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-06 14:41:18.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-05-06 16:25:36.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-05-06 16:38:15.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-05-06 16:44:23.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-06 17:57:11.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-05-06 18:29:53.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-08 00:57:52.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-11 21:24:03.000000000","updated_by":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"REVIEWER"},{"updated":"2020-05-13 08:27:29.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 12:54:43.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2020-07-29 11:09:26.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 11:41:34.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 11:45:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-29 12:03:39.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 12:08:06.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-07-29 13:04:33.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 13:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 13:48:46.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 14:05:08.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-07-29 17:11:14.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 21:18:49.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 21:29:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-29 22:02:19.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-07-29 23:56:26.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-07-30 03:55:21.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 09:48:48.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 13:00:14.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 12:53:38.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-19 14:04:44.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 14:07:18.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-02-19 14:21:37.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 15:56:04.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-02-19 16:12:51.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-02-19 19:45:32.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-20 04:32:08.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-20 05:51:24.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-02-21 09:19:10.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-07-05 09:41:16.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"decb1e169df97159c711ce0d6de81fad116dbbcb","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-04-22 17:02:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4cdc962707383927366fca742f18bab831f5900","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-04-22 17:04:13.000000000","message":"Topic set to bug/1860913","accounts_in_message":[],"_revision_number":1},{"id":"2850670fb241d829eb04541029e01d605ff5b782","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-22 17:35:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/81/722081/1/check/DellEMC_PowerMAX_FC/dddab21/DellEMC_PowerMAX_FC/697 : FAILURE in 18m 18s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/81/722081/1/check/DellEMC_PowerMAX_iSCSI/f78885f/DellEMC_PowerMAX_iSCSI/594 : FAILURE in 16m 22s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"66903595058fc9ca4178b48bf1586a317d7b9cd1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-22 17:38:52.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/81/722081/1/check-not-vote/ext-cinder-zuul/05bda78 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a181d74c8b8030df76c2ceaec5e1c8940c4f29ff","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-22 17:40:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/81/722081/1/check/fujitsu-eternus-dx-iscsi/b97cb91 : FAILURE in 37m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/81/722081/1/check/fujitsu-eternus-dx-fc/9c896fa : FAILURE in 37m 29s","accounts_in_message":[],"_revision_number":1},{"id":"00f96c683cdd5db0b5e8d05a61a5c26232609042","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-22 17:46:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/81/722081/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/39e8460/39e8460.html : FAILURE in 11m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ebcc901032ea8c7f89b1fd71975262bed5dacb7d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-22 17:55:07.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722081/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1d049d6261e5a1bffe8e51e90843c391cf9a0e48","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-22 18:01:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722081/1/iscsi/ : SUCCESS in 58m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/722081/1/smb/ : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f4394e0461e1c9d51e75c08431edade0cdb44159","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-22 18:17:31.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/81/722081/1/check/DellEMC_XtremIO_FC/12b82c6/DellEMC_XtremIO_FC/1139 : FAILURE in 16m 24s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/81/722081/1/check/DellEMC_XtremIO_iSCSI/ddb64b3/DellEMC_XtremIO_iSCSI/1030 : FAILURE in 16m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"5049395e81c14f6450159f29c959824a11537913","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-22 18:18:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/81/722081/1/check/DellEMC_Unity_FC/b116f27/DellEMC_Unity_FC/1032 : FAILURE in 26m 41s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/81/722081/1/check/DellEMC_Unity_iSCSI/3867fd1/DellEMC_Unity_iSCSI/984 : FAILURE in 17m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f9b62ea02ac5f01a2e2cfb58f000567c5dcc7fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 20:39:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fa2e12633a6443bb9683095caca4545 : FAILURE in 24m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/747caf443c2a48d8908874de770c9c6f : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/917e6181e16b43788c2197ad29a8b483 : SUCCESS in 10m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a88a5c254f4414c802edc4b3f6685e4 : SUCCESS in 9m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c816ed6404314f089fa49f4c7fa017c8 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baac89596a2a440db260166b21071171 : SUCCESS in 14m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/92cfd4d1b1d14932a024da4723a7e04b : SUCCESS in 1h 13m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/861a85446b6e422fba424dc5290998d4 : SUCCESS in 1h 36m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2753ac1543ae4a19ac4ba0e168530c5d : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4a4d11b9f7e4f0c82890e65561bc541 : SUCCESS in 7m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e7e8dd481bb642feb8a06a9ab6ec393c : SUCCESS in 8m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e9be603044e946e09bb9a8b8c933e0c7 : SUCCESS in 40m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/17e004314fa44a7c95cdc9e0bb277787 : SUCCESS in 7m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9320dfc550cf4e559e7542154146341e : SUCCESS in 1h 08m 38s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e4a7ff73ee5f473b9535fe642b9791c0 : SUCCESS in 1h 46m 59s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f352ba1a3a0f46b09587231100d516b3 : SUCCESS in 1h 17m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/387fef38bbe64cf996d90b698c95246a : SUCCESS in 1h 45m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9e05657b40b143d886d07b43c1e93be7 : SUCCESS in 1h 37m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e6c577fc9f34577b78e7f735b2bf5a7 : SUCCESS in 2h 13m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d019c7d674cf4eb2a4d0f8319b861215 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"28abb683b2418baae485382da01377914938e771","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-22 20:55:10.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-722081-1/ns5_nfs : FAILURE in 3h 44m 25s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-722081-1/ns5_iscsi : FAILURE in 3h 52m 11s","accounts_in_message":[],"_revision_number":1},{"id":"25952cc9817517a9d9ed7381999188d78fd50f4d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-22 21:17:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/81/722081/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3360a50 : SUCCESS in 55m 12s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/81/722081/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b05a41f : SUCCESS in 1h 04m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"5b866124e717d9a61b7857ba6dcb0a421c435eb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-22 22:33:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"233f4b191db9860d2a77fb07d74409b44dc1d5dc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-22 23:00:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/1/check/tempest-dsvm-full-bionic/e1970f9/ : SUCCESS in 1h 40m 46s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/1/check/devstack-plugin-ceph-dsvm-bionic/6601326/ : SUCCESS in 1h 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"47f506ed72fac70d33cd5352997c42c72932a954","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-22 23:13:27.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-722081-1/ns5_nfs : FAILURE in 33m 21s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-722081-1/ns5_iscsi : FAILURE in 38m 40s","accounts_in_message":[],"_revision_number":1},{"id":"450465412a2d0823a14c98c6f6f6b17a1d5a6055","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-22 23:44:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/81/722081/1/check/nimble-iscsi-driver-dsvm-volume/5d41db1 : SUCCESS in 1h 21m 45s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/81/722081/1/check/nimble-fc-driver-dsvm-volume/0209166 : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7a46470c241ca325e90bfd0aad2819eb2c7f4fe0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-23 02:36:34.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/1/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d9d1747dfb9c971b8c4a045c9c5a2d1b0bb08e6e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-23 02:44:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/81/722081/1/check/dsvm-tempest-cisco-zonemanager-job/9d24ba2/ : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b0d65ad747f8b1c047d8197bd654b5baabf703ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-23 03:29:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/92/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8396a498a4e4e3c231103b733a9f7bb79aeef8f5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-23 11:37:22.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-722081-1/ns5_nfs : FAILURE in 1h 57m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-722081-1/ns5_iscsi : FAILURE in 2h 1m 29s","accounts_in_message":[],"_revision_number":1},{"id":"37a0729272651c1c2fedab178e4b293565c9c73c","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-23 16:08:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/805/index.html : FAILURE in 30m 26s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/954/index.html : SUCCESS in 1h 36m 43s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/993/index.html : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1adb74c43bc30ed4894e5f4698d317e1c247a63e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-24 01:53:49.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-722081-1/ns5_nfs : SUCCESS in 2h 2m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-722081-1/ns5_iscsi : SUCCESS in 2h 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"bf3a2d89734b8ec71ba2bcb6c51f5b51c9c4d9db","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-04-24 06:04:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/1/20200424122002/NEC-ISCSI : SUCCESS in 2h 41m 08s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/1/20200424122002/NEC-FC : SUCCESS in 2h 44m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e00c753930b5125f09c546d8376eae2312cbee87","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-24 08:20:05.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d93efacaa8cdc4692f6c6bf3ceaa3b86479aeae8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-24 10:04:37.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"97a31b0d129893863d7a1c4173d0ae997d6037d9","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-04-24 10:42:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73403bdf3f57481dcbfcc794b901989f2f20d5f3","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-24 11:34:40.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722081/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"93bf4aa8e9c8e7952c826c2b3d48ac4802466dd4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-24 11:42:22.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722081/2/iscsi/ : FAILURE in 59m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/722081/2/smb/ : FAILURE in 58m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3de8381f786d92ba13720fddb5d7c5e3c7322da4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-24 12:35:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/81/722081/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e9b554b : FAILURE in 1h 37m 28s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/81/722081/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/856580c : SUCCESS in 1h 14m 46s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"87e0172621e0f41b2f5443cf37d79932d6a9b6e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-24 12:37:05.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/112/index.html : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"60f055fe052651137c16415af390cdee5a2fff0f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-24 12:42:07.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-722081-2/ns5_nfs : SUCCESS in 1h 59m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-722081-2/ns5_iscsi : FAILURE in 30m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f1543e823a024f2472f0f5e5fa7a42fae1ca3fbd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-24 12:49:11.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/81/722081/2/check-not-vote/ext-cinder-zuul/8312cff : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8f7e95581914f120234d512a7c9794dc6f7d192f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:50:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53749e658e6c46f4866dce86170c5a60 : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76cf55a6724f4590a3ee27a3d8cbd7a6 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e1e1b7f90e345b594694d000de29941 : SUCCESS in 11m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af97c2165c1a499abf139ac84bec3314 : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ad7315b44624785b730c2df965f239b : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e547dc6cadeb4dfc96b281c47e9a6c33 : SUCCESS in 12m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cb0d0ec8012f4a329d14cb367f5f7f93 : SUCCESS in 1h 17m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06eb683e37c246a3ba25f49bcb296ff7 : SUCCESS in 1h 17m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8977897ce89492585cab5e15ed227fa : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3aad1e642e64204a390d2f1c6a9fab8 : SUCCESS in 6m 06s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/508db732934e4ab3a253419ebb139eb8 : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/da1b1b11f3b84b8b9f89ba5381cb0f93 : SUCCESS in 38m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bae0be7830c04089a1785f8ad2c7be59 : SUCCESS in 4m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/23f6347fbbce463abbe23a81c0477b39 : SUCCESS in 1h 02m 28s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/95f47762763248f39f5e97c71009c20e : SUCCESS in 1h 50m 52s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/39fce6a0480c40e6921488d73dea438b : SUCCESS in 1h 21m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/635488fb1f54466fbb9aeaae45785a21 : FAILURE in 1h 38m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/40b5135b66034f8d8930c1b41cf230a9 : SUCCESS in 1h 57m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/98bdb8ad5558444485dccf536cb750cb : FAILURE in 3h 06m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d1c5d0f2b8d42e99d0402f564d34759 : SUCCESS in 1h 30m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8a59c8c7be43fc430e1ace04a08669a62d3d0795","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-24 14:28:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/81/722081/2/check/nimble-iscsi-driver-dsvm-volume/178a01d : FAILURE in 1h 54m 17s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/81/722081/2/check/nimble-fc-driver-dsvm-volume/47e1ee0 : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":2},{"id":"786d71f3e757a839bd31b6db7adb8c383e68d407","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-24 14:41:32.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/2/check/tempest-dsvm-full-bionic/a9982f8/ : SUCCESS in 1h 34m 15s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/2/check/devstack-plugin-ceph-dsvm-bionic/c9fbbcb/ : SUCCESS in 2h 01m 51s","accounts_in_message":[],"_revision_number":2},{"id":"758496a1972613e131d2e629d45232153670dd15","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-24 15:04:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/81/722081/2/check/fujitsu-eternus-dx-iscsi/ae31100 : FAILURE in 48m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/81/722081/2/check/fujitsu-eternus-dx-fc/1dc10de : FAILURE in 36m 07s","accounts_in_message":[],"_revision_number":2},{"id":"871926e712b833b51ae4d7ae442e20c9b8b0949e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-24 16:15:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/81/722081/2/check/dsvm-tempest-cisco-zonemanager-job/53684f1/ : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":2},{"id":"de01d5d76ab3e096dc368d5e12289e596ee510bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-24 17:04:43.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/2/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"15d9087cd3a30bbcbcdfb2ea2d892747d2efc1b5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-25 01:25:41.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/81/722081/2/check/DellEMC_Unity_FC/404c5f8/DellEMC_Unity_FC/1061 : SUCCESS in 1h 27m 06s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/81/722081/2/check/DellEMC_Unity_iSCSI/23c6498/DellEMC_Unity_iSCSI/1011 : SUCCESS in 1h 51m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c4df3e8e556bd87fedf1487f9dd1714cac27cf76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-25 03:45:44.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/2/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f23517ef91e5ffee334dd39d8ddf0a64cc7b450d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-25 14:47:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/81/722081/2/check/DellEMC_XtremIO_FC/29a060c/DellEMC_XtremIO_FC/1166 : SUCCESS in 1h 33m 28s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/81/722081/2/check/DellEMC_XtremIO_iSCSI/b746af6/DellEMC_XtremIO_iSCSI/1057 : FAILURE in 2h 23m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"1b3d514e4205af47bc63c9d5113789fae4767019","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-25 15:07:48.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/2/Cinder-lioadm : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"bd875f9160d44617e109efc09483a89e966ba73f","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-25 19:19:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/832/index.html : SUCCESS in 1h 14m 31s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/981/index.html : SUCCESS in 1h 41m 29s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1020/index.html : SUCCESS in 1h 23m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6ea118cc8f32a53e8b2b0b92f10c992fade649a0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-27 12:34:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/81/722081/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/5ca9163/5ca9163.html : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7670f349afc4b88f7151f7797f5e30ba9249b771","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-27 13:06:00.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"76bb89d0240693eb47579c214afb0c0e683db649","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-27 21:02:15.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-722081-2/ns5_nfs : SUCCESS in 4h 17m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-722081-2/ns5_iscsi : SUCCESS in 4h 39m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a9861417e950afd42eb276420f94653f48f81bca","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-27 21:32:04.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"89be2ddc2b52912a14efb02218dea5222c990dcb","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-04-29 10:28:48.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/2/20200429152809/NEC-ISCSI : FAILURE in 1h 50m 10s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/2/20200429152809/NEC-FC : SUCCESS in 4h 00m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0ce0a1555845af72bf029c2171010ed1f5ea3a4a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-29 11:19:38.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/2/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e17cedb2a8de429dc85378b8a4dc4d0bfee7046d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-29 20:17:09.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"c5d0d8cfce358b530f499a7274e144a46b50ec53","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-05-06 10:28:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"78434871453070da44755b21a621a314c22775b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 10:34:37.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"fb7e37ade0d5695e779365585559c95e8ff75347","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 11:02:02.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/dd75485513a5b33497c4dd44aa47a5d81c32978c/index.html : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"69b57ef3e1b3e0df42d126f95008caf6d4a208d1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-05-06 11:03:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/81/722081/3/check/fujitsu-eternus-dx-iscsi/a5e5fb0 : FAILURE in 34m 45s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/81/722081/3/check/fujitsu-eternus-dx-fc/6d20c9d : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":3},{"id":"f35cd2e0d2a341c8f0c60c2cc0c8a2e756b4f253","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-05-06 11:05:38.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/81/722081/3/check-not-vote/ext-cinder-zuul/14400ed : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":3},{"id":"3c4fc3b9308c9a9ba8483aac3ee5e05e8b86c559","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 11:09:40.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"203db77777abf9bdfd4dbe58efb04e4dae24a680","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-05-06 11:13:00.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/81/722081/3/check/cinder-storpool-tempest/e8d7f3d/ : RETRY_LIMIT in 12m 34s","accounts_in_message":[],"_revision_number":3},{"id":"08a87a151099de3983735aba79d0ccbfcb30bb0e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-05-06 11:18:45.000000000","message":"Patch Set 3:\n\nIn summary, the problem is with improper refresh of the connection info, i\u0027ve few questions regarding this:\n\n1) Is this issue specific to nfs or other *fs drivers as well (like quobyte, smbfs etc), since the original issue was reported against glusterfs\n2) (not sure how relevant this is) but from the details on launchpad, after the snapshot creation, the original volume file freezes and instance starts writing to the snapshot file. IIUC snapshot is a point in time copy of the state, shouldn\u0027t the original volume file be written to and snapshot file should be kept intact?","accounts_in_message":[],"_revision_number":3},{"id":"8a270ded36393434c4198707c55615cdb08bc39b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-05-06 11:20:09.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722081/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9d3d9b93a870a8d18b6a397ae3a15d99715abf10","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-05-06 11:33:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/81/722081/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/3c5f17b/3c5f17b.html : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":3},{"id":"612a0291dc9f00d1582bd482236abb7ff4da12e8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-05-06 11:36:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722081/3/iscsi/ : SUCCESS in 1h 03m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/722081/3/smb/ : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":3},{"id":"9afeca070b2bdbc2629d65f9fa9c0284e9d1a60d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 12:22:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/81/722081/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/fe684a1 : SUCCESS in 1h 02m 12s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/81/722081/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/21eaf7c : FAILURE in 1h 53m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"f251c625952d044274d13ce3b0ebb4acb36cdea8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-05-06 12:23:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 45m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a89dd34fc26ec7a187f9a06b233be3c55071eea4","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-05-06 12:29:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/81/722081/3/check/nimble-iscsi-driver-dsvm-volume/69cbcab : SUCCESS in 1h 27m 41s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/81/722081/3/check/nimble-fc-driver-dsvm-volume/e1de5e7 : SUCCESS in 1h 21m 03s","accounts_in_message":[],"_revision_number":3},{"id":"2c75d0860a04f02775e12687d21afcfa2000b62e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-06 12:36:00.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-722081-3/ns5_nfs : SUCCESS in 2h 3m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-722081-3/ns5_iscsi : SUCCESS in 2h 6m 58s","accounts_in_message":[],"_revision_number":3},{"id":"2d5c8aefc0f1188a78237140f4e0c7c771836e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:51:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce8f9a3c80c489a988705206c354791 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7725f2870d43bea940836e32f4f0cc : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26bde268ca34464a8997ba2f1ada7390 : SUCCESS in 9m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd00b33c4e2a4167ad9c47a921a2a222 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507b78bdebd24395a437e8231c451223 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/0fb1d6af914a4222be266d676dc9ff64 : SUCCESS in 1h 01m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/661eedcfb7c642cb9fcc3a1b04ce14de : SUCCESS in 1h 19m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11da5be453f84e20b004a360cc134c9f : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3ed68bc0cf44739aa2fc8776d61e41f : SUCCESS in 7m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60bfacbb01bb4e3d81e882ea522c2dbc : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7afdf4efbd324484b35c106de3c7d1a3 : SUCCESS in 44m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fadbc77a59214817b952e31e100cc48b : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f3d51e497b844b07b3794419f1853caa : SUCCESS in 1h 06m 52s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7bc0fcf596fc4fc28bc909d7e0e80dd0 : SUCCESS in 1h 21m 44s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9303ee062f954f0aaea7038ca7e8cfd1 : SUCCESS in 1h 08m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/96395a27bd424a7282a897cf48ab036e : SUCCESS in 1h 22m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c36479258dfa4ec0b5b785795f721a20 : SUCCESS in 1h 16m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c08b113e7cfc4e76bb7d5d5ca87e4a18 : SUCCESS in 2h 17m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/54d14fdf48db40a69b9326c12ec03f46 : SUCCESS in 1h 11m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddf5e1483d89476caf7e46581a24dfc2 : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":3},{"id":"2cc0ea9e13de6fab7ec735bc884750eec613d7d2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-06 13:16:55.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/3/check/tempest-dsvm-full-bionic/7224144/ : SUCCESS in 1h 55m 55s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/3/check/devstack-plugin-ceph-dsvm-bionic/a78551a/ : FAILURE in 2h 05m 36s","accounts_in_message":[],"_revision_number":3},{"id":"4df14e663b3f3470546b5aabed59840d83fd0146","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 13:22:58.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c5a102fa6f5c9f3dbc9344cfc9652cb2e2fd217b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-05-06 13:54:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/81/722081/3/upstream-solidfire-check/cinder-solidfire-iSCSI/fdf0cdb/ : SUCCESS in 1h 01m 27s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"7a20085f382b1bfd516ec980e9e758e882692bfa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-06 14:02:47.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-722081-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"4c7d921ac4aa7e163efd2b2a7db11162bd9507f2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-06 14:04:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/81/722081/3/check/dsvm-tempest-infinibox-fc/9cf4f58 : SUCCESS in 1h 14m 15s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/81/722081/3/check/dsvm-tempest-infinibox-iscsi/3b1be8c : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":3},{"id":"056caa165e4884a14a662e0b844f9d3b587e2b34","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-05-06 14:05:45.000000000","message":"Patch Set 3:\n\n\u003e In summary, the problem is with improper refresh of the connection\n \u003e info, i\u0027ve few questions regarding this:\n\nThere\u0027s no problem here for the NFS c-vol backend, rather the opposite. It is safe to make repeat calls to the initialize_connection API with this backend however the caller has no idea this is the case.\n\n \u003e 1) Is this issue specific to nfs or other *fs drivers as well (like\n \u003e quobyte, smbfs etc), since the original issue was reported against\n \u003e glusterfs\n\nI\u0027ve not looked tbh but I\u0027d assume so yes.\n\n \u003e 2) (not sure how relevant this is) but from the details on\n \u003e launchpad, after the snapshot creation, the original volume file\n \u003e freezes and instance starts writing to the snapshot file. IIUC\n \u003e snapshot is a point in time copy of the state, shouldn\u0027t the\n \u003e original volume file be written to and snapshot file should be kept\n \u003e intact?\n\nThat would avoid the need to refresh the connection_info but I can\u0027t think of a way to switch to that layout *and* keep things backwardly compatible for existing volumes with snapshots.","accounts_in_message":[],"_revision_number":3},{"id":"0175b9631da20cf9dd17c15228786dcb48078cc7","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-05-06 14:41:18.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/722081-3/Lenovo-iSCSI : FAILURE in 19:37.63\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/722081-3/Lenovo-FC : FAILURE in 19:54.89","accounts_in_message":[],"_revision_number":3},{"id":"a3f073848c948145849e02108153ffb69c1aea58","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 14:48:20.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/81/722081/3/check/DellEMC_XtremIO_FC/eac8df7/DellEMC_XtremIO_FC/1270 : SUCCESS in 1h 36m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/81/722081/3/check/DellEMC_XtremIO_iSCSI/f3d4f13/DellEMC_XtremIO_iSCSI/1157 : SUCCESS in 1h 38m 57s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"0fbc2e0a0869068c870b0fa5ffe33e4e39d35418","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 14:54:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"94e2fea58cb67aa29d721a506db9f0a496c62032","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-06 15:19:48.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/81/722081/3/check/DellEMC_Unity_FC/4082f29/DellEMC_Unity_FC/1232 : SUCCESS in 1h 29m 54s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/81/722081/3/check/DellEMC_Unity_iSCSI/4a39e40/DellEMC_Unity_iSCSI/1107 : SUCCESS in 1h 46m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c1083082a703870668b76724bbc117118b5c59b4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 15:25:36.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"93b89bbc4d2be3ee0117b7f12e9efe86b50dee63","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-05-06 16:25:36.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/722081-3/Seagate-iSCSI : SUCCESS in 48:38.95\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/722081-3/Seagate-FC : SUCCESS in 47:42.18","accounts_in_message":[],"_revision_number":3},{"id":"6278de20ffe769873d4217dccf4d342cb56054c5","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-05-06 16:38:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/941/index.html : SUCCESS in 1h 14m 01s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1089/index.html : FAILURE in 17s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1124/index.html : FAILURE in 1h 45m 22s","accounts_in_message":[],"_revision_number":3},{"id":"ea2a6b0551044bd97434bad985edf7d44f1714e8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 16:44:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/81/722081/3/upstream-check/cinder-cDOT-iSCSI/d37de93/ : SUCCESS in 47m 20s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/81/722081/3/upstream-check/cinder-cDOT-iSCSI-IPv6/3fcb902/ : SUCCESS in 2h 56m 27s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/81/722081/3/upstream-check/cinder-cDOT-NFS/e200bf3/ : FAILURE in 3h 51m 38s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/81/722081/3/upstream-check/cinder-cDOT-NFS-IPv6/82d90d2/ : SUCCESS in 2h 28m 32s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/81/722081/3/upstream-check/cinder-cDOT-FCP/b544f0e/ : FAILURE in 3h 49m 57s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"787744d34c40743f6ff87f99cbcd5109a5f08206","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-06 17:38:24.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"abbaa0e9feaf44f152a7df15762510fa809abb13","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-05-06 17:57:11.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/81/722081/3/check/DellEMC_SC_FC/1d37e8a/DellEMC_SC_FC/747 : FAILURE in 1h 10m 34s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/81/722081/3/check/DellEMC_SC_iSCSI/57a0e4c/DellEMC_SC_iSCSI/745 : FAILURE in 2h 51m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"9a0e51d3e7036aac6d0087c374a9008f62d993cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 18:03:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"e04545fad256bcdd4d034358153e45fa4dbc556c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 18:29:53.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/81/722081/3/check/DellEMC_VNX_iSCSI/a7022d1/DellEMC_VNX_iSCSI/630 : SUCCESS in 2h 50m 19s\n- DellEMC_VNX_FC http://publiclogs.emc.com/81/722081/3/check/DellEMC_VNX_FC/ef26336/DellEMC_VNX_FC/633 : SUCCESS in 57m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"732e0b015402e027b9afa438a2b9953cea0d5192","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-06 18:34:18.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"a4b3b6b857fed27e42747505c38b16a1d78331de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 18:35:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"f63ba6e80ab0e8db2febd908a0af3887305687f1","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-05-08 00:57:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/81/722081/3/check/DellEMC_PowerMAX_FC/582f83c/DellEMC_PowerMAX_FC/753 : SUCCESS in 8h 22m 36s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/81/722081/3/check/DellEMC_PowerMAX_iSCSI/3386e69/DellEMC_PowerMAX_iSCSI/634 : SUCCESS in 9h 46m 44s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"62a516d7a5535fef14aed45d37e838183ca9fc03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-08 01:03:33.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"793292c0f817f2e1b66fda2dff815ec1ec48e935","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-08 22:14:31.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"015749a6b7b1f64084e51cc270d93a7f9cf17012","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2020-05-11 21:24:03.000000000","message":"Patch Set 3:\n\nrun-storpoolci","accounts_in_message":[],"_revision_number":3},{"id":"6b8ccd99cc8eda5404e7a2965ddff25e2816ca04","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-05-13 08:27:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/81/722081/3/check/cinder-storpool-tempest/1c8e638/ : SUCCESS in 2h 34m 45s","accounts_in_message":[],"_revision_number":3},{"id":"dffa116a672879083fca1c3c31f79c07eb5b8e96","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-29 10:50:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5bf7c99ed121b4ce7f7ce6fbc21ad3e0907400e4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-29 11:09:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/81/722081/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/b18c838/b18c838.html : FAILURE in 6m 05s","accounts_in_message":[],"_revision_number":4},{"id":"7f5f0cf0d879439e30789cc82e0ae2e2bd3a1c09","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-29 11:41:34.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722081/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"6e186441eba22188cf68440b5ffc179e258b49aa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 11:45:27.000000000","message":"Patch Set 4:\n\nThere was an effort done a few years back to go through all the drivers and update them to allow multiple calls to initialize_connection once we realized nova was (ab)using it that way.\n\nIt can make sense to report this behavior explicitly, but if we are going to do that, then we need to add this for all Cinder drivers, not just NFS.","accounts_in_message":[],"_revision_number":4},{"id":"e8f8009e827d8d67f5dc6e094067164265736821","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-29 12:03:39.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/81/722081/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/af73d2e : FAILURE in 20m 38s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/81/722081/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/1033e28 : FAILURE in 20m 44s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"56485046e213e7402d3dc2874c5c1e69ef5890aa","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-29 12:08:06.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722081/4/iscsi/ : FAILURE in 53m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/722081/4/smb/ : FAILURE in 1h 17m 08s","accounts_in_message":[],"_revision_number":4},{"id":"0a0adb7f3e0d28002c76e6ff50c45ea7c05767d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-29 12:09:19.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"d96d63e156dd482c78f432aeb1aae900cc650835","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-29 13:04:33.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-722081-4/ns5_nfs : SUCCESS in 2h 6m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-722081-4/ns5_iscsi : SUCCESS in 2h 13m 40s","accounts_in_message":[],"_revision_number":4},{"id":"03f5a21114eb028fe19456334e097476983e6f9f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-29 13:11:42.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"b1098974f87c818067ab24dbd74b82b3942b219d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 13:41:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/227607d5cf3847e6b931efd1a40c5428 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bf72a4a809846b5b6d44af19c772a11 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09bb1dee98864674899a29de60d385e0 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/419032d237ea46c8965febfb8749b7bf : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d124344f64284c6a81b59c99ea234e42 : SUCCESS in 17m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/dd902b485d91427cbfc0bf20ff5c6c9b : SUCCESS in 54m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/000952d023884dd6b2b56bc86bf5fbc3 : SUCCESS in 1h 49m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce26dd0637ed42f5a0ebf9fdf3595256 : SUCCESS in 12m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/982aff84fe524d6fb588f4848e010a11 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9738dd07a4e4169a9d7225a0c32f5cc : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2674fa519214611b01fa25c8be1a706 : SUCCESS in 5m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4f67c74bd1ca45d2b4908654da82f8e6 : SUCCESS in 39m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71d4a14203ea421589017bd902eaee84 : SUCCESS in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/430bb9c3d5d54f9c950f094ccccdfc78 : SUCCESS in 1h 16m 05s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/862f1cdb998e4b6faa78832413b86e50 : FAILURE in 1h 23m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ab663f98e1084cc3939d00017af8f2d1 : SUCCESS in 1h 23m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0afe95223e5646a39cb977de58690cf6 : SUCCESS in 1h 32m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a7c631b6ee014807b7fe851cdbee0e3d : SUCCESS in 1h 19m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f336b08a7fa64134b0187202935045cc : SUCCESS in 1h 26m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9b8352cea36410d8751d2f90bf7f2e0 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":4},{"id":"058f2950ba1aeaa2374b8ce69ba2e1b6a4581e7b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-29 13:44:24.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"206a9bde58d9bf1dbadd088674a18e4fe957a6cb","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-29 13:48:46.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/b4546a332be2e796ef5c40e26a1c8bed46f7cc9f/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"fbbcb9243ba8f99c85ce2c82584cd3b717b179f1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-29 14:05:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/81/722081/4/check/dsvm-tempest-cisco-zonemanager-job/560e757/ : FAILURE in 34m 26s","accounts_in_message":[],"_revision_number":4},{"id":"9c42f37bce7b0983a8cb03649e778533de0acf98","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-29 16:53:15.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"171745099cd8f9a4766498746329618c5d0a7c27","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-29 17:11:14.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/4/check/tempest-dsvm-full-bionic/175b877/ : SUCCESS in 1h 36m 22s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/4/check/devstack-plugin-ceph-dsvm-bionic/6900b39/ : SUCCESS in 1h 44m 49s","accounts_in_message":[],"_revision_number":4},{"id":"d5fb7d168f37f82b487a5c39ef61b5faeb17cedb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-29 18:36:31.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"3b2fa3ff6dcb6b83e9b684baa7b92d0ab8af5bd8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-29 21:18:49.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/81/722081/4/check/DellEMC_XtremIO_FC/0cb9d4c/DellEMC_XtremIO_FC/528 : SUCCESS in 1h 34m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/81/722081/4/check/DellEMC_XtremIO_iSCSI/e4c949b/DellEMC_XtremIO_iSCSI/501 : SUCCESS in 1h 38m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"dca974835bde51559ed3aa4d49d3b0638260518f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-29 21:29:22.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/81/722081/4/check/DellEMC_Unity_FC/0056e31/DellEMC_Unity_FC/336 : SUCCESS in 1h 14m 01s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/81/722081/4/check/DellEMC_Unity_iSCSI/503091b/DellEMC_Unity_iSCSI/326 : SUCCESS in 1h 40m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"99faa7261261728379151e36454b2b332e6898c1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-29 22:02:19.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/722081/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3b285049adff533ae67c6ce6b7f1604d5f32c81c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-07-29 23:56:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/81/722081/4/check/dsvm-tempest-infinibox-fc/b4b689a : SUCCESS in 1h 08m 04s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/81/722081/4/check/dsvm-tempest-infinibox-iscsi/a6cda3f : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":4},{"id":"00fa56473e1cbb7d143f613ef71fa7e6d7884d18","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-30 03:55:21.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"264159ab679b77a3ff9dd5dab55ee7bb183ff6f8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 09:48:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/4/20200730162304/NEC-FC : SUCCESS in 2h 18m 19s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/4/20200730162304/NEC-ISCSI : SUCCESS in 2h 25m 42s","accounts_in_message":[],"_revision_number":4},{"id":"bd251cfde2faeb2d9156bd2dd1d466d752be3efe","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-30 13:00:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/81/722081/4/check/nimble-iscsi-driver-dsvm-volume/06654a7 : SUCCESS in 2h 35m 24s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/81/722081/4/check/nimble-fc-driver-dsvm-volume/9c75ec3 : SUCCESS in 2h 08m 13s","accounts_in_message":[],"_revision_number":4},{"id":"4e10e5a643b705edd9fb3492a133c553a9363304","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-07 22:07:16.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"28c016aaf2d0f516446b6baedc1ab9e38303690d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-07 22:57:03.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/722081/722081/5/check-not-vote/ext-cinder-zuul/f6c657c : FAILURE in 48m 59s","accounts_in_message":[],"_revision_number":5},{"id":"71eb49e7197a8cd3efcaceaffbc7a7ee0af3f75e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-07 23:03:15.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/a0a8d00accaf4cee93399bdd6907f741 : DISK_FULL in 46m 29s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e131f7cd83bd4efa91fe21c6a9bada27 : DISK_FULL in 41m 12s","accounts_in_message":[],"_revision_number":5},{"id":"03a3ec3a9b9fbfda9a78643e87d9948305856710","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-07 23:22:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722081/5/iscsi/ : SUCCESS in 1h 14m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/722081/5/smb/ : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":5},{"id":"0fe0b8f862a25900675264af193394680a483fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-07 23:27:28.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/81/722081/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ae10b56/ae10b56.html : FAILURE in 1h 18m 03s","accounts_in_message":[],"_revision_number":5},{"id":"4d14e2c548ca60a960b69abab8f764cb920a483f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-07 23:32:55.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/81/722081/5/check/DellEMC_Unity_FC/cb6f111/DellEMC_Unity_FC/250 : SUCCESS in 1h 25m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/81/722081/5/check/DellEMC_Unity_iSCSI/716404d/DellEMC_Unity_iSCSI/230 : FAILURE in 30m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"f16ae4e61178181b933d82dc95ce586baad8c0ba","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-08 00:35:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/81/722081/5/check/dsvm-tempest-cisco-zonemanager-job/7b1a8ef/ : SUCCESS in 1h 14m 59s","accounts_in_message":[],"_revision_number":5},{"id":"674e82892dcedd4fe8c5a3b0b9d200b8e68116af","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-08 00:57:38.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/81/722081/5/check/DellEMC_XtremIO_FC/85d894e/DellEMC_XtremIO_FC/265 : SUCCESS in 1h 43m 06s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/81/722081/5/check/DellEMC_XtremIO_iSCSI/717c9ed/DellEMC_XtremIO_iSCSI/264 : SUCCESS in 1h 44m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"18a90a6a7f80a9c75eec89ba3b921747a51dffbd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-08 01:13:51.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-722081-5/ns5_nfs : SUCCESS in 3h 3m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-722081-5/ns5_iscsi : SUCCESS in 3h 5m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"1f8acb05becf09351b6a2f60dedb46981121f0a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 01:24:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0599a3456ce47e9a6be63361d1fa946 : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f43e23136d046e4ab3f589288a86077 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3a525501154f29a3e1c0a264ef00aa : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d74fa6f4c8a43f9a65b74e684d81877 : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f6ff2c58e344938aecfa9bb1958a7de : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ec530ea7274e3ea842613193e4c577 : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/abc0210eb46949418c29e37404b8bc1d : SUCCESS in 1h 10m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb86ae03dcce4d7a9888bbec3c258499 : SUCCESS in 1h 13m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/498df8abdcc246cbb009bde8254af60c : SUCCESS in 11m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3512d5f3ce23437699dff0dc7b5ff49b : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26ea5ca4fa7347e686664a223bc18563 : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/14bc9f31ebc74c04a5c1dbadc43dbb4d : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3947a88271d04c7fb430b9d415df4d38 : SUCCESS in 50m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2d1a5fc786cb49158b4094b02a3db697 : FAILURE in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e4d2df4c975c4605bbf54db1770b8764 : SUCCESS in 1h 19m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/26414cefcfd84780b2db90c1d0a1e7d5 : FAILURE in 1h 26m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/847feec87baa4ba6b2272e04a31f2f11 : SUCCESS in 1h 45m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8f5bf116f9a4f40babd9dce9f52136f : SUCCESS in 1h 09m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ff4a489b7424ca89f9c781b48443f69 : TIMED_OUT in 3h 15m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab66430dbd4e4eeebb65b8829a659269 : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/adaf54e0db5c4688a347794cce181db6 : SUCCESS in 2h 16m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3066f7a604a48929c6e90b4e0a62d7e : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":5},{"id":"2b06371dd1e16b7818178400b59cf3c70c6f97d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-08 02:31:38.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/5/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"b16dc8a2277c041a0c33f9482a510f83a7396d7a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-08 05:45:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/5/20201208123409/NEC-FC : SUCCESS in 2h 01m 22s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/5/20201208123409/NEC-ISCSI : SUCCESS in 2h 11m 35s","accounts_in_message":[],"_revision_number":5},{"id":"be72cadd2cf2b6b86eb436d147b1515e3036090f","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-12-08 10:46:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1977/index.html : SUCCESS in 1h 20m 44s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/2123/index.html : FAILURE in 1h 14m 27s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/2112/index.html : FAILURE in 54m 26s","accounts_in_message":[],"_revision_number":5},{"id":"c0a63222fb0c5519f9b26087a7ac4c0fac2d9674","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-09 15:51:19.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"a1487bf231c32398b1ab0d3fd7fc5d42eeb54ec5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-02-19 12:50:09.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"c3ba691125374b8eb425705cd9d591f47b4d5658","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-19 12:53:38.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/81/722081/6/check/dsvm-tempest-TOYOU-ACS5000-driver/54fcaec : FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"82cbd760136c5bb8a24626889f7399dfe3f5aaff","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-19 12:55:29.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722081/6/iscsi/ : FAILURE in 2m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/722081/6/smb/ : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":6},{"id":"ea7dde8b801f928ab974e90e2c0318d3844dca88","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-19 13:31:09.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/722081/722081/6/check-not-vote/ext-cinder-zuul/d0379ec : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":6},{"id":"be752c6fb725e42dc04dd103c121ad889efa6a14","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-19 13:55:11.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/81/722081/6/check/iscsi-driver-primera/b0c0eb1/b0c0eb1.html : FAILURE in 1h 03m 02s","accounts_in_message":[],"_revision_number":6},{"id":"6e092c678a7160457d7f0d4c5d746ffb8836d52a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-19 14:03:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/81/722081/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/850b5fb/850b5fb.html : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":6},{"id":"999399ad96aa55e0acdde2c1cdd8754764f6d9d2","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-19 14:04:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/81/722081/6/check/dsvm-tempest-zadara-driver/7072099 : SUCCESS in 1h 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1ebc39c7de2605acd988470a6a3256c7b2c0dc61","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-19 14:07:18.000000000","message":"Patch Set 6: Code-Review+2\n\nVery small change.  The previous check failures appear to be unrelated.  Just need it to pass the check.","accounts_in_message":[],"_revision_number":6},{"id":"4b1dd76e57c695f34b9d632f018a947d5af55e01","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-19 14:11:13.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/81/722081/6/check/dsvm-tempest-cisco-zonemanager-job/4c6bb87/ : SUCCESS in 1h 15m 48s","accounts_in_message":[],"_revision_number":6},{"id":"4a3c2b5b0b4e38dc8d0689716b0b4fd7ac6f50b6","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-19 14:17:17.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM, definitely need to call the correct API to fetch or refresh connection_info.","accounts_in_message":[],"_revision_number":6},{"id":"f6b67471781587e9f81631aeed21676ff986b72e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-19 14:21:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/81/722081/6/ : FAILURE in 1h 11m 01s","accounts_in_message":[],"_revision_number":6},{"id":"0272595ac06fcdef92ddbf42beab662f83153b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:14:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/884150c1923e4211889b1e5b975174a0 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b0cf99a6abd46d5a700494a2493f47a : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf05d1eabe7a4a53a3074128791000b0 : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d969c7b351154f5686c4418793558e1e : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f96d9e56f9a04ac29b78319032a4de64 : SUCCESS in 14m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c2fd99ea244d49b5db83b4cdaf01b9 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/adab77ba6a0648b895c8ee37e22e303e : SUCCESS in 1h 10m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8d4e444d1cbf406db50559adc6f2b231 : SUCCESS in 1h 34m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b8f5ec0fbb55430f877d6f4ed21b8e96 : FAILURE in 13m 35s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8d825cf0e1e4f51a26dd9c5ff481b36 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa4fdf3c3adb4a3489d228ba75e464bc : SUCCESS in 6m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73b6c9461ac84b3a99e20607b854b60e : SUCCESS in 6m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/caa7513c4105413e9a6792c4e81bc7c2 : SUCCESS in 46m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/462b9ee3fd58426e80f59c3e79d230e7 : SUCCESS in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b3e450f493ac4d308ac651dacc96e521 : SUCCESS in 1h 57m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6ea79a3983064b7db99a2d93b4b8215b : FAILURE in 1h 26m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3b8c0e6dd59743d8a3bb01f4f4b75d17 : SUCCESS in 1h 23m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/28caf99720cb4a6abc55e3bfb8652459 : SUCCESS in 1h 13m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6c022ff8889a45a2ba5ee53dd307dfa4 : SUCCESS in 1h 27m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0fb3129e614c44e8aca5b561f8f1f7a7 : SUCCESS in 1h 36m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de2b6d0143f34c55a06cd2eab8a28141 : SUCCESS in 2h 18m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49cada82d3c64ea291ca1df7e851fa89 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":6},{"id":"3702f79466d401088decb274e9b89fb2c7ecf7b4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-19 15:56:04.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/722081/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/99230ea/ : TIMED_OUT in 3h 04m 41s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/722081/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/af42d6c/ : SUCCESS in 1h 45m 53s","accounts_in_message":[],"_revision_number":6},{"id":"e9f467e1453e05b764cf517e09af6f9443aa46c8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-19 16:12:51.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/81/722081/6/check/DellEMC_PowerStore_iSCSI/55b6153/DellEMC_PowerStore_iSCSI/449 : SUCCESS in 57m 06s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/81/722081/6/check/DellEMC_PowerStore_FC/633ae16/DellEMC_PowerStore_FC/443 : SUCCESS in 49m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"0add9a067ed2981d6eb358d4bfa96a74c3c660e7","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-19 16:20:26.000000000","message":"Patch Set 6: Code-Review-1\n\nRegarding https://review.opendev.org/c/openstack/cinder/+/722659/2#message-1397946dd6580e27186fb74fdda0a7cc7870db7d I wonder why nfs\u0027s unittest didn\u0027t fail. \n\nDoes this means that we should add a \u0027test_initialize_connection\u0027 test for NFS?\nhttps://opendev.org/openstack/cinder/src/branch/master/cinder/volume/drivers/nfs.py#L98","accounts_in_message":[],"_revision_number":6},{"id":"79f442a4f9ab2ca89f27ec4f95754e112f73c738","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 16:42:22.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/81/722081/6/check/DellEMC_VNX_iSCSI/37156d0/DellEMC_VNX_iSCSI/419 : FAILURE in 1h 26m 36s\n- DellEMC_VNX_FC http://publiclogs.emc.com/81/722081/6/check/DellEMC_VNX_FC/15fdc28/DellEMC_VNX_FC/438 : FAILURE in 57m 50s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"860d2a410517c52262c4f6be52be4b2e278ac806","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-19 16:49:03.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/81/722081/6/upstream-solidfire-check/cinder-solidfire-iSCSI/e9a5166/ : SUCCESS in 1h 32m 16s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"f8f8f558d2e96a42ef80d4d5a98f1df292b58524","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 17:40:20.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"400195b0add9c493d1aa28bcec501a3f5344131d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-19 18:11:40.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/81/722081/6/upstream-check/cinder-cDOT-iSCSI/45bef8e/ : SUCCESS in 40m 20s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/81/722081/6/upstream-check/cinder-cDOT-iSCSI-IPv6/6b0d705/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/81/722081/6/upstream-check/cinder-cDOT-NFS/e61a1af/ : SUCCESS in 38m 34s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/81/722081/6/upstream-check/cinder-cDOT-NFS-IPv6/db3485a/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/81/722081/6/upstream-check/cinder-cDOT-FCP/979d086/ : FAILURE in 2h 55m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"1df1e048c62d39e11a64d4d81140ee448b3a7099","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 19:45:32.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/81/722081/6/check/DellEMC_PowerFlex/57aefa7/DellEMC_PowerFlex/417 : FAILURE in 2h 29m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"dd329b241cdd5364a6782480c054e8d01cfc6bfc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-19 23:22:48.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/6/check/tempest-dsvm-full-focal-py3/e2d1595/ : SUCCESS in 1h 53m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/6/check/devstack-plugin-ceph-dsvm-focal-py3/a03cca3/ : TIMED_OUT in 2h 39m 23s","accounts_in_message":[],"_revision_number":6},{"id":"80632bfe48cca9f6a4b6a672119672eb32203293","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-20 03:32:03.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"aa0b46932fee31c9bccd671fb5db2eeb3b918dba","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-20 04:32:08.000000000","message":"Patch Set 6:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/722081-6/PVME-iSCSI : SUCCESS in 50:20.41\n* PVME-FC http://os-logs.tristero.net/dell-ci/722081-6/PVME-FC : SUCCESS in 47:51.77","accounts_in_message":[],"_revision_number":6},{"id":"323ae76035afa1df4d53578f7a2d24e5e1d902af","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-02-20 05:51:24.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b6d71b63ff03dad2eb80a49070604984a158780c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-02-20 05:54:28.000000000","message":"Removed Code-Review+1 by wushiming \u003cwushiming@yovole.com\u003e\n","accounts_in_message":[],"_revision_number":6},{"id":"d633cd8dca6e30e12d5c4fe0c76ccfab40a7d3d9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-20 12:40:40.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-722081-6/ns5_nfs : FAILURE in 59m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-722081-6/ns5_iscsi : SUCCESS in 6h 58m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"c5c44f2be826d6f2b88ed950b4dc72471bde7d7a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-02-20 16:47:44.000000000","message":"Patch Set 6:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/722081-6/Seagate-iSCSI : SUCCESS in 47:30.27\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/722081-6/Seagate-FC : SUCCESS in 45:37.32","accounts_in_message":[],"_revision_number":6},{"id":"c628802ac4652a8a4190a8f6ab9beee18a55f20a","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-20 20:27:54.000000000","message":"Patch Set 6:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/722081-6/Lenovo-iSCSI : SUCCESS in 46:04.76\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/722081-6/Lenovo-FC : SUCCESS in 46:40.08","accounts_in_message":[],"_revision_number":6},{"id":"0b75b0e292bb2ce56268858dfa71da9a3ce8eeca","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-20 21:16:45.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/81/722081/6/check/DellEMC_XtremIO_FC/3a3d0ca/DellEMC_XtremIO_FC/815 : SUCCESS in 2h 03m 16s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/81/722081/6/check/DellEMC_XtremIO_iSCSI/4972a9e/DellEMC_XtremIO_iSCSI/817 : SUCCESS in 1h 38m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"00c2189c6cdfb1d956ec3e7d67d57f85dec576d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-20 23:01:43.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"b6488842e243fd096468ad598696da0b8aac2ead","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-21 09:19:10.000000000","message":"Patch Set 6:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/722081-6/HPMSA-iSCSI : FAILURE in 1:04:33\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/722081-6/HPMSA-FC : SUCCESS in 43:20.61","accounts_in_message":[],"_revision_number":6},{"id":"b59b39f77f7fc031b5a22e8cf0d76c4f8da43d48","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-22 02:42:20.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-722081-6/ns5_nfs : SUCCESS in 3h 53m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-722081-6/ns5_iscsi : SUCCESS in 3h 45m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"615c19ab214bf3d03dfa0dd8409f85ca9dd37b48","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-22 04:11:00.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"ec2fc64a9b0c992e0348328618cf35bef1f77b8d","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-22 22:40:08.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/6/20210223032817/NEC-ISCSI : SUCCESS in 2h 08m 35s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/6/20210223032817/NEC-FC : SUCCESS in 4h 11m 49s","accounts_in_message":[],"_revision_number":6},{"id":"cd46c34665f69a0a5a1d85912b28391bfa1f97f8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-26 07:17:22.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/81/722081/6/check/DellEMC_PowerMAX_FC/be6458d/DellEMC_PowerMAX_FC/504 : SUCCESS in 5h 23m 58s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/81/722081/6/check/DellEMC_PowerMAX_iSCSI/f3ff9ac/DellEMC_PowerMAX_iSCSI/469 : SUCCESS in 5h 03m 03s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"d7e241a59a781748df892daed97f72175624542e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-26 16:19:45.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/81/722081/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"9c9cb4aa1a4d8d76529b89dfed5ebc384499fbf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 17:14:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"537cf726740b7e6055f0ea94980bc237bd70eb58","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-18 17:20:55.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/81/722081/7/check/dsvm-tempest-TOYOU-ACS5000-driver/44badfc : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":7},{"id":"cebb65d2876dd6a85f3603b6b21eef965de32bef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-18 17:57:13.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/722081/722081/7/check-not-vote/ext-cinder-zuul/830d431 : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":7},{"id":"85834052dc2c52ae8c39a8c04e302d8ba0bdcbd0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-18 18:30:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/81/722081/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/525d278/525d278.html : FAILURE in 1h 14m 16s","accounts_in_message":[],"_revision_number":7},{"id":"4fdc667429f62b55808bf12b0fc420fe0ca10c23","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-18 18:32:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/81/722081/7/check/iscsi-driver-primera/87d2e71/87d2e71.html : FAILURE in 1h 16m 23s","accounts_in_message":[],"_revision_number":7},{"id":"14706995a079db4aa619e5bacb78930d8e97e5d1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-18 19:35:37.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/722081/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c87e8ec/ : POST_FAILURE in 2h 19m 20s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/722081/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9eaf9f9/ : SUCCESS in 2h 09m 58s","accounts_in_message":[],"_revision_number":7},{"id":"9d244f676fda65b0381947f1d83e204929eeebb2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-18 20:01:04.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/81/722081/7/ : FAILURE in 1h 27m 50s","accounts_in_message":[],"_revision_number":7},{"id":"81230ee66ec16b7769ef4e85345f45ad4c5a9f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 20:12:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1452c0b2fe13427096c236c19182c62d : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dbb1644b6634b2c908600798ab24564 : SUCCESS in 12m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ab6588c90554e00920fb1d1bed1811c : SUCCESS in 15m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3204df2825e408cb4db4e80759c171d : SUCCESS in 14m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66dab47d47a940d2b0985afdff17ac8a : SUCCESS in 14m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2241b7be6a7345b9873188396bcfe9ca : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/7aa092cdbaff4e278508bbb983761d7c : SUCCESS in 1h 06m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ccb53f79d998414991c31814dbd5be43 : SUCCESS in 1h 45m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3a7c4816e50b443da98849b027be63a5 : SUCCESS in 13m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/57646a83ecab43e4a97e20ba41302565 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a14ee1554db6469c8c7b75b461cdb5e3 : SUCCESS in 6m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f47abbeec6e245c59dbb40409f99fe0c : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/85d4f064beae4cf4889bc22ea2370367 : SUCCESS in 40m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6d3a1a5a27a5403fa9b17cf16b3fc45e : FAILURE in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4833ae42ec7d456093fc093cab84218f : SUCCESS in 1h 22m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/687a1fc12c2c48b68f2affa9cd9f4878 : FAILURE in 1h 25m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/12da1c6c088943c0a51398a038ed8375 : SUCCESS in 1h 51m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/000ef282067848d9876f567926f8974a : SUCCESS in 1h 14m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3fa71e832a924f669d1d2436e1cdef0e : SUCCESS in 1h 25m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f6b22dd5071a4367afcd43dfba9e618f : FAILURE in 1h 45m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91bb88989b1548cf88bd540dd260559f : SUCCESS in 2h 06m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7f6debedc7f484d91c79e2e4fc89b2b : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":7},{"id":"b360f92076dcfe94b077535e6f02f610de9f52e1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-18 21:14:14.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/81/722081/7/upstream-check/cinder-cDOT-iSCSI/d53f715/ : SUCCESS in 42m 42s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/81/722081/7/upstream-check/cinder-cDOT-iSCSI-IPv6/71395ff/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/81/722081/7/upstream-check/cinder-cDOT-NFS/5c80cdd/ : SUCCESS in 40m 47s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/81/722081/7/upstream-check/cinder-cDOT-NFS-IPv6/2e392fe/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/81/722081/7/upstream-check/cinder-cDOT-FCP/8044009/ : SUCCESS in 1h 00m 32s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"9a4ce9f2e62e1d9addee449aaeba0ee2f39cf041","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-18 21:21:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/81/722081/7/check/dsvm-tempest-zadara-driver/7e53784 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e3ccc42e638a646c898f3c7e50ab62db8891e1a9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-18 21:27:03.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/7/check/tempest-dsvm-full-focal-py3/3f5d5a1/ : SUCCESS in 2h 09m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/722081/7/check/devstack-plugin-ceph-dsvm-focal-py3/4e3e573/ : TIMED_OUT in 2h 39m 59s","accounts_in_message":[],"_revision_number":7},{"id":"306a6ab770bc17b3fd5ab17537de3bf1d8827d6a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-18 22:48:49.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/81/722081/7/check/DellEMC_PowerStore_iSCSI/33b9d8f/DellEMC_PowerStore_iSCSI/647 : SUCCESS in 52m 03s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/81/722081/7/check/DellEMC_PowerStore_FC/95a431d/DellEMC_PowerStore_FC/641 : SUCCESS in 54m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"0e591df8b73ebea7250a234c8910ea0b79959ccf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-18 22:51:55.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/81/722081/7/check/dsvm-tempest-cisco-zonemanager-job/5a8f337/ : FAILURE in 1h 22m 13s","accounts_in_message":[],"_revision_number":7},{"id":"eae88a64ac63da5bed3b6815db13bad50864ddd1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-18 23:00:07.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/81/722081/7/upstream-solidfire-check/cinder-solidfire-iSCSI/6525019/ : FAILURE in 2h 46m 06s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"0cddc4a897bf370886d4ed5c226c373df9bab17a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-19 01:12:36.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/722081-7/Seagate-iSCSI : SUCCESS in 45:26.15\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/722081-7/Seagate-FC : SUCCESS in 40:42.12","accounts_in_message":[],"_revision_number":7},{"id":"c66e7b15fe1831ce603595e24f68aceaad4077c7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-19 10:55:22.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/81/722081/7/check/DellEMC_VNX_iSCSI/d2aafb6/DellEMC_VNX_iSCSI/594 : SUCCESS in 3h 09m 18s\n- DellEMC_VNX_FC http://publiclogs.emc.com/81/722081/7/check/DellEMC_VNX_FC/b9372af/DellEMC_VNX_FC/632 : SUCCESS in 1h 19m 35s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"570a29c61f9e35076503ec19a1a955b7b4301f53","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-19 18:24:51.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/81/722081/7/check/DellEMC_SC_FC/6ca480d/DellEMC_SC_FC/462 : SUCCESS in 2h 00m 18s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/81/722081/7/check/DellEMC_SC_iSCSI/97372c4/DellEMC_SC_iSCSI/459 : SUCCESS in 2h 08m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"0fb77e198a0108d4eb0e9c077668eb6b718aee06","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-20 14:50:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/722081/7/check/iscsi-driver-nimble/586bb4c/586bb4c.html : FAILURE in 1h 16m 09s","accounts_in_message":[],"_revision_number":7},{"id":"9e6596d1e421e4da3bba0aabae2915e86a902a06","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-20 18:34:23.000000000","message":"Patch Set 7:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/722081-7/PVME-iSCSI : SUCCESS in 1:01:52\n* PVME-FC http://os-logs.tristero.net/dell-ci/722081-7/PVME-FC : SUCCESS in 56:34.45","accounts_in_message":[],"_revision_number":7},{"id":"5fd8ac8b93380302034e402f1b7ae4c9b54da4bb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-20 19:52:40.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/81/722081/7/check/DellEMC_PowerFlex/9fdfe7b/DellEMC_PowerFlex/614 : FAILURE in 2h 47m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"2ca29e2dc7a16e3c83088214c40e691662268f77","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-20 23:33:40.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-722081-7/ns5_nfs : SUCCESS in 5h 6m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-722081-7/ns5_iscsi : SUCCESS in 4h 58m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"de8f0d25574b53d83ce2da70f60492e848a390b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-21 20:50:59.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/81/722081/7/check/DellEMC_Unity_FC/e996b59/DellEMC_Unity_FC/922 : SUCCESS in 1h 36m 23s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/81/722081/7/check/DellEMC_Unity_iSCSI/af88c21/DellEMC_Unity_iSCSI/874 : SUCCESS in 1h 53m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"a1da4b747c829ec481488649f5a5a20799c3ec3b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 05:13:21.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/81/722081/7/check/DellEMC_XtremIO_FC/3f301fe/DellEMC_XtremIO_FC/1087 : SUCCESS in 2h 03m 09s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/81/722081/7/check/DellEMC_XtremIO_iSCSI/1e6d1f1/DellEMC_XtremIO_iSCSI/1091 : SUCCESS in 1h 48m 31s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"dbe750393f00f4d3a19210373beeb7baabdda7d6","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 18:07:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/7/20210323004123/NEC-ISCSI : SUCCESS in 2h 13m 15s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/722081/7/20210323004123/NEC-FC : SUCCESS in 2h 26m 19s","accounts_in_message":[],"_revision_number":7},{"id":"8c8a5a515056d35f6ab58bfb0531f73572423002","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-07-05 09:41:16.000000000","message":"Patch Set 7: Code-Review-1\n\nIn my opinion we should not add this information to the connection information.\nThis is internal Cinder driver information that should **not** be exposed to normal users (an admin could be a different thing).\nAny Cinder user/consumer should assume drivers are not idempotent and act accordingly.\n\nIf other cores disagree, we should evaluate adding this as a driver class attribute (or instance attribute if idempotency is based on storage array/firmware) with a safe default (False) and it\u0027s the manager that sets the field instead of the driver for both, connection_information and driver capabilities.","accounts_in_message":[],"_revision_number":7},{"id":"12ecbae2f5fa3369494a60c9c54b036e9abbcd66","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-05 16:27:37.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7: Code-Review-1\n\u003e \n\u003e In my opinion we should not add this information to the connection information.\n\u003e This is internal Cinder driver information that should **not** be exposed to normal users (an admin could be a different thing).\n\u003e Any Cinder user/consumer should assume drivers are not idempotent and act accordingly.\n\u003e \n\u003e If other cores disagree, we should evaluate adding this as a driver class attribute (or instance attribute if idempotency is based on storage array/firmware) with a safe default (False) and it\u0027s the manager that sets the field instead of the driver for both, connection_information and driver capabilities.\n\nThat\u0027s fine but at this point I\u0027ll be throwing bug #1860913 back over to the cinder folks to refresh the connection_info associated with an attachment during cinder.volume.drivers.remotefs.RemoteFSSnapDriverBase._{create,delete}_snapshot_online as without something like this in connection_info n-cpu isn\u0027t going to be able to determine what is and isn\u0027t safe to call here.","accounts_in_message":[],"_revision_number":7},{"id":"0ea7a59b7ce6a90f1a94f2b8df55cf6f64207f01","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-07-05 16:48:07.000000000","message":"Patch Set 7:\n\nThe newer attachment api was supposed to address this. This call was never meant to be idempotent, then we discovered nova was calling it relatedly to get this info. Isn\u0027t nova supposed to be making the new calls now?","accounts_in_message":[],"_revision_number":7},{"id":"ce36f61327145ac2c869a35276c7a6ffcffe418b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-05 16:58:53.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e The newer attachment api was supposed to address this. This call was never meant to be idempotent, then we discovered nova was calling it relatedly to get this info. Isn\u0027t nova supposed to be making the new calls now?\n\nIt\u0027s currently making a GET against the attachment and using the connection_info it provides.\n\nThe issue here was that the connection_info we were getting back hadn\u0027t been updated by the remotefs driver when creating or deleting snapshots so I wanted to update (PUT) the attachment with a fresh connector to get this recreated.\n\nTalking to Gorka it seems like he is on-board with simply fixing the remotefs driver so it updates the connection_info in the attachment *before* n-cpu does the GET.","accounts_in_message":[],"_revision_number":7},{"id":"21d241605d1531ba17a6f3b7cb306edbd4faacd8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-05 17:19:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9d3e5aae651d2bc5277ece42b8622ae82927380e","revisions":{"a58caf8607ea88d1b09ab28d211a968c0d09b0fe":{"kind":"REWORK","_number":1,"created":"2020-04-22 17:02:38.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/81/722081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/722081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/722081/1"}}},"commit":{"parents":[{"commit":"e9d9050e86cb17654fa7897c5864c1fc20828e69","subject":"Merge \"Fix driver doc ordering do to file casing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9d9050e86cb17654fa7897c5864c1fc20828e69"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-22 16:41:39.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-22 17:02:35.000000000","tz":60},"subject":"DNM/WIP nfs: Mark the initialize_connection API as idempotent in the connection_info","message":"DNM/WIP nfs: Mark the initialize_connection API as idempotent in the connection_info\n\nChange-Id: I2dd06c43c39ddb576fea9e5cbccffc096e77073a\nRelated-bug: #1860913\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a58caf8607ea88d1b09ab28d211a968c0d09b0fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a58caf8607ea88d1b09ab28d211a968c0d09b0fe"}]},"branch":"refs/heads/master"},"343f11172786fa20cf9cb0a5c91eb02a1ec0ff5f":{"kind":"REWORK","_number":2,"created":"2020-04-24 10:42:08.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/81/722081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/722081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/722081/2"}}},"commit":{"parents":[{"commit":"e9d9050e86cb17654fa7897c5864c1fc20828e69","subject":"Merge \"Fix driver doc ordering do to file casing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9d9050e86cb17654fa7897c5864c1fc20828e69"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-22 16:41:39.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-24 10:32:59.000000000","tz":60},"subject":"nfs: Mark the initialize_connection API as idempotent in connection_info","message":"nfs: Mark the initialize_connection API as idempotent in connection_info\n\nThere are unfortunately no guarantees that calls to the\ninitialize_connection API provided by each volume driver are idempotent.\nAs a result callers have no idea if it is ever safe to make repeat calls\nto the API when attempting to refresh connection_info.\n\nThis change adds a simple initialize_idempotent key to the\nconnection_info returned by the NFS volume driver to indicate that\nrepeat calls to the initialize_connection for this driver are indeed\nsafe. This should then allow callers such as n-cpu to correctly call the\ncorrect API to fetch (attachment show) or refresh (attachment update)\nconnection_info.\n\nRelated-bug: #1860913\nChange-Id: I2dd06c43c39ddb576fea9e5cbccffc096e77073a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/343f11172786fa20cf9cb0a5c91eb02a1ec0ff5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/343f11172786fa20cf9cb0a5c91eb02a1ec0ff5f"}]},"branch":"refs/heads/master"},"dd75485513a5b33497c4dd44aa47a5d81c32978c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-06 10:28:40.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/81/722081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/722081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/722081/3"}}},"commit":{"parents":[{"commit":"e82ff57b70bb94c39c0eacf9d32a38a411571581","subject":"Merge \"Monkey patch original current_thread _active\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e82ff57b70bb94c39c0eacf9d32a38a411571581"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-22 16:41:39.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-05-06 10:28:40.000000000","tz":0},"subject":"nfs: Mark the initialize_connection API as idempotent in connection_info","message":"nfs: Mark the initialize_connection API as idempotent in connection_info\n\nThere are unfortunately no guarantees that calls to the\ninitialize_connection API provided by each volume driver are idempotent.\nAs a result callers have no idea if it is ever safe to make repeat calls\nto the API when attempting to refresh connection_info.\n\nThis change adds a simple initialize_idempotent key to the\nconnection_info returned by the NFS volume driver to indicate that\nrepeat calls to the initialize_connection for this driver are indeed\nsafe. This should then allow callers such as n-cpu to correctly call the\ncorrect API to fetch (attachment show) or refresh (attachment update)\nconnection_info.\n\nRelated-bug: #1860913\nChange-Id: I2dd06c43c39ddb576fea9e5cbccffc096e77073a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd75485513a5b33497c4dd44aa47a5d81c32978c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd75485513a5b33497c4dd44aa47a5d81c32978c"}]},"branch":"refs/heads/master"},"b4546a332be2e796ef5c40e26a1c8bed46f7cc9f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-29 10:50:28.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/81/722081/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/722081/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/722081/4"}}},"commit":{"parents":[{"commit":"2124f39f9d54bc0db1e1a3bf526441b1729196a2","subject":"Merge \"Bump hacking version to 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2124f39f9d54bc0db1e1a3bf526441b1729196a2"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-22 16:41:39.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-07-29 10:50:28.000000000","tz":0},"subject":"nfs: Mark the initialize_connection API as idempotent in connection_info","message":"nfs: Mark the initialize_connection API as idempotent in connection_info\n\nThere are unfortunately no guarantees that calls to the\ninitialize_connection API provided by each volume driver are idempotent.\nAs a result callers have no idea if it is ever safe to make repeat calls\nto the API when attempting to refresh connection_info.\n\nThis change adds a simple initialize_idempotent key to the\nconnection_info returned by the NFS volume driver to indicate that\nrepeat calls to the initialize_connection for this driver are indeed\nsafe. This should then allow callers such as n-cpu to correctly call the\ncorrect API to fetch (attachment show) or refresh (attachment update)\nconnection_info.\n\nRelated-bug: #1860913\nChange-Id: I2dd06c43c39ddb576fea9e5cbccffc096e77073a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4546a332be2e796ef5c40e26a1c8bed46f7cc9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4546a332be2e796ef5c40e26a1c8bed46f7cc9f"}]},"branch":"refs/heads/master"},"b728e92e7a84eedfea6c949740ec9b113fc407b8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-12-07 22:07:16.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/81/722081/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/722081/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/722081/5"}}},"commit":{"parents":[{"commit":"95d00bd7a6658e2cd3066e5ae77d3a9a63bc0c8f","subject":"Merge \"Remove collections.abc backwards compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/95d00bd7a6658e2cd3066e5ae77d3a9a63bc0c8f"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-22 16:41:39.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-07 22:07:16.000000000","tz":0},"subject":"nfs: Mark the initialize_connection API as idempotent in connection_info","message":"nfs: Mark the initialize_connection API as idempotent in connection_info\n\nThere are unfortunately no guarantees that calls to the\ninitialize_connection API provided by each volume driver are idempotent.\nAs a result callers have no idea if it is ever safe to make repeat calls\nto the API when attempting to refresh connection_info.\n\nThis change adds a simple initialize_idempotent key to the\nconnection_info returned by the NFS volume driver to indicate that\nrepeat calls to the initialize_connection for this driver are indeed\nsafe. This should then allow callers such as n-cpu to correctly call the\ncorrect API to fetch (attachment show) or refresh (attachment update)\nconnection_info.\n\nRelated-bug: #1860913\nChange-Id: I2dd06c43c39ddb576fea9e5cbccffc096e77073a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b728e92e7a84eedfea6c949740ec9b113fc407b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b728e92e7a84eedfea6c949740ec9b113fc407b8"}]},"branch":"refs/heads/master","description":"Rebase"},"f612dff47789f66a8f3dab400ee97ceb112e2ac8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-02-19 12:50:09.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/81/722081/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/722081/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/722081/6"}}},"commit":{"parents":[{"commit":"367e704614d8901759eb4abd4bc1aea0dc42bbc0","subject":"Merge \"Update to hacking 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/367e704614d8901759eb4abd4bc1aea0dc42bbc0"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-22 16:41:39.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-02-19 12:50:09.000000000","tz":0},"subject":"nfs: Mark the initialize_connection API as idempotent in connection_info","message":"nfs: Mark the initialize_connection API as idempotent in connection_info\n\nThere are unfortunately no guarantees that calls to the\ninitialize_connection API provided by each volume driver are idempotent.\nAs a result callers have no idea if it is ever safe to make repeat calls\nto the API when attempting to refresh connection_info.\n\nThis change adds a simple initialize_idempotent key to the\nconnection_info returned by the NFS volume driver to indicate that\nrepeat calls to the initialize_connection for this driver are indeed\nsafe. This should then allow callers such as n-cpu to correctly call the\ncorrect API to fetch (attachment show) or refresh (attachment update)\nconnection_info.\n\nRelated-bug: #1860913\nChange-Id: I2dd06c43c39ddb576fea9e5cbccffc096e77073a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f612dff47789f66a8f3dab400ee97ceb112e2ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f612dff47789f66a8f3dab400ee97ceb112e2ac8"}]},"branch":"refs/heads/master","description":"Rebase"},"9d3e5aae651d2bc5277ece42b8622ae82927380e":{"kind":"REWORK","_number":7,"created":"2021-03-18 17:14:44.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/81/722081/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/722081/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/722081/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/722081/7"}}},"commit":{"parents":[{"commit":"a66f688992dae0a7bd26cdfc866af539be15ebca","subject":"Merge \"Support format info in fs type drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a66f688992dae0a7bd26cdfc866af539be15ebca"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-22 16:41:39.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 17:14:38.000000000","tz":0},"subject":"nfs: Mark the initialize_connection API as idempotent in connection_info","message":"nfs: Mark the initialize_connection API as idempotent in connection_info\n\nThere are unfortunately no guarantees that calls to the\ninitialize_connection API provided by each volume driver are idempotent.\nAs a result callers have no idea if it is ever safe to make repeat calls\nto the API when attempting to refresh connection_info.\n\nThis change adds a simple initialize_idempotent key to the\nconnection_info[\u0027data\u0027] dict returned by the NFS volume driver to\nindicate that repeat calls to the initialize_connection for this driver\nare indeed safe. This should then allow callers such as n-cpu to\ncorrectly call the correct API to fetch (attachment show) or refresh\n(attachment update) connection_info.\n\nA NOTE and FIXME are left detailing a wrinkle when using the cinderv3\nattachment update API that results in this key and mount_point_base\nending up in the nested data dict that means this key has been added\ndirectly to this dict while strictly not belonging there.\n\nRelated-bug: #1860913\nChange-Id: I2dd06c43c39ddb576fea9e5cbccffc096e77073a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d3e5aae651d2bc5277ece42b8622ae82927380e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d3e5aae651d2bc5277ece42b8622ae82927380e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
