)]}'
{"id":"openstack%2Fcinder~722655","triplet_id":"openstack%2Fcinder~master~If63b90e80b38c1995ddf364520be2dec34ed9dfd","project":"openstack/cinder","branch":"master","topic":"backup-config","hashtags":[],"change_id":"If63b90e80b38c1995ddf364520be2dec34ed9dfd","subject":"WIP. Refactor backup configuration","status":"NEW","created":"2020-04-24 10:26:03.000000000","updated":"2020-10-27 21:13:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":60,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2543e4b1fee9354a08e371a53f49b8ef9c7ec1e","_number":722655,"virtual_id_number":722655,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},{"date":"2020-10-02 15:33:03.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"date":"2020-10-12 19:23:43.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2020-10-12 22:03:49.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2020-10-12 19:36:36.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"date":"2020-10-02 15:16:18.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-02 21:02:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-27 21:13:31.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"date":"2020-10-12 22:40:28.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"date":"2020-10-02 20:19:25.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"date":"2020-10-04 13:25:42.000000000","_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"date":"2020-10-03 07:21:37.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"date":"2020-10-02 16:28:28.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"date":"2020-10-14 12:31:21.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"date":"2020-10-14 08:47:56.000000000","_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"date":"2020-10-02 21:20:13.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},"all":[{"value":1,"date":"2020-10-02 17:32:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-25 13:43:10.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-28 14:43:37.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"REVIEWER"},{"updated":"2020-05-06 16:27:00.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-08-28 13:24:42.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-08-30 21:17:21.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-09-12 07:51:24.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 14:47:46.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 15:08:33.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-09-18 18:57:21.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 19:15:40.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-10-02 15:16:18.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-02 15:33:03.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-02 16:28:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-02 17:32:25.000000000","updated_by":{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},"reviewer":{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},"state":"REVIEWER"},{"updated":"2020-10-02 20:19:25.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-02 21:02:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-02 21:20:13.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-03 07:21:37.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-04 13:25:42.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-12 19:23:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-10-12 22:03:49.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-12 22:40:28.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-10-14 08:47:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 12:31:21.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-27 21:13:31.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d97e8c39cd166dbc6f1e7bff18741695e4f8802","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-24 10:26:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ec690b5a0db378ae52cfbe9fd89738d9920311c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-24 11:00:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/55/722655/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/dca75ec : FAILURE in 28m 32s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/55/722655/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/986c601 : FAILURE in 29m 56s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"a014a12fb1951677b49c44b6c15d7bd7abd67acd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-24 11:06:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/111/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9a10548de248379532fac4cefa587ed655a6150a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-24 11:09:33.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-1/ns5_nfs : FAILURE in 42m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-1/ns5_iscsi : FAILURE in 43m 6s","accounts_in_message":[],"_revision_number":1},{"id":"85854b162c6b842aa2fc3a417cc9488247f086ca","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-24 11:10:57.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722655/1/iscsi/ : FAILURE in 44m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/722655/1/smb/ : FAILURE in 44m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cdae2bdf4b1bd302eabc6a3bf36aa1ad28f43b38","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-24 11:18:20.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3a4f9806476c8a98f98005cd0a8886caa1b2ef54","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-24 12:41:00.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/55/722655/1/check-not-vote/ext-cinder-zuul/c736b2d : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"feb12b81a34130fc91932012055a2b56b42c4ea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 12:42:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/096ae6a179c84b488182bb60aac259a7 : FAILURE in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a72b293dd6274154be8abde7f480a573 : FAILURE in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/484a62baa50a43798aef0b1350e7c8cb : FAILURE in 9m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39590d221d7045bbac5866f47fb11f3a : FAILURE in 16m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec9d22d537234642b9023273cd1273e2 : FAILURE in 16m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abdd51a3b40a4a698848cbb4dc8f462f : SUCCESS in 17m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/de83f11b26f74e8ab3e8bfa8f0a95dbd : SUCCESS in 1h 07m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9a938561e1c415aaf14004e183962d6 : FAILURE in 31m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a51cafdb432c4569b0cd5e5b2a4124fa : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/944d72c790cb41fb873125bd1fc5b834 : SUCCESS in 6m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/aeed3042ca4b462e992f884c3a94f9c1 : SUCCESS in 6m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/692cdcad908a43388c6240075c251f3c : FAILURE in 29m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de1e878580eb4965887f5e702142bc3b : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/acf553a3e3964f1fbeb61a58e408ad8d : SUCCESS in 1h 12m 12s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c2c0341424844ffe977f170b03e22e26 : FAILURE in 30m 14s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ff90bb7cec3b43668fed572752afc966 : SUCCESS in 1h 15m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0b197c553f974de485b9df55982d58ec : FAILURE in 27m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e891e914d1f400891e6f49adcb23e07 : FAILURE in 1h 05m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5fbc0f487afb4b96b68a11e46078a570 : FAILURE in 2h 13m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eb92f2b47024053acf1fc63fedad646 : FAILURE in 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5e44952d4a1c2a24ae1b3f101d54b51845e9df29","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-24 13:13:27.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/55/722655/1/check/nimble-iscsi-driver-dsvm-volume/6d72814 : FAILURE in 1h 52m 30s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/55/722655/1/check/nimble-fc-driver-dsvm-volume/4016a79 : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2a8f937ab4c8474b55daec459adac151b0ce5fcc","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-24 14:13:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/55/722655/1/check/fujitsu-eternus-dx-iscsi/361aff1 : FAILURE in 42m 37s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/55/722655/1/check/fujitsu-eternus-dx-fc/5a2af87 : FAILURE in 49m 31s","accounts_in_message":[],"_revision_number":1},{"id":"19a6c13c871a3d6a3a9b6db68b7da18d92c90672","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-24 14:34:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/55/722655/1/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/55/722655/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"17f5ab396ff6a40bf1121211ae0627fad1703be9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-24 15:07:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/55/722655/1/check/dsvm-tempest-cisco-zonemanager-job/33efd6b/ : FAILURE in 52m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c790581138f0d439dbeb6eac707fc230c7475b37","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-24 18:51:17.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/1/check/tempest-dsvm-full-bionic/3c173a2/ : SUCCESS in 1h 57m 36s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/1/check/devstack-plugin-ceph-dsvm-bionic/d7e71f5/ : FAILURE in 1h 21m 44s","accounts_in_message":[],"_revision_number":1},{"id":"bdcbfa54aebe17bf817f4f412633419aee84c4de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-24 21:58:06.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/55/722655/1/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/55/722655/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c00bfff67958761945831d27653765bfb68a6f14","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-25 12:23:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/55/722655/1/check/DellEMC_XtremIO_FC/8d0df48/DellEMC_XtremIO_FC/1165 : FAILURE in 36m 18s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/55/722655/1/check/DellEMC_XtremIO_iSCSI/230c7e1/DellEMC_XtremIO_iSCSI/1056 : FAILURE in 30m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"08eb19e7d2ad4f0ecea142d0abaa86742d4ad17f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-25 13:43:10.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/55/722655/1/Cinder-tgtadm : ABORTED\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/55/722655/1/Cinder-lioadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"df8f81a9b7030cf4d9980cf7448f81900b5f04d6","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-25 16:41:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/831/index.html : FAILURE in 33m 15s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/980/index.html : FAILURE in 43m 10s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1019/index.html : FAILURE in 36m 13s","accounts_in_message":[],"_revision_number":1},{"id":"046c4f6267b37efd93286a9f3e5352da93c58df7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-27 11:04:40.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"98b66d7dcf9f16d4e4d07a3b20282d3f41dbc5f7","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-27 14:50:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e7116e61894bec4e447b1b5bc1d3a5d6c7b3510","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-27 15:16:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/78946d5/78946d5.html : POST_FAILURE in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0340132c2bcfb7d4af2cd41c088feb829f54297b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-27 15:20:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"36b7336595227d81fb16d7a42eeff4436757a989","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-27 15:23:42.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-2/ns5_nfs : FAILURE in 32m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-2/ns5_iscsi : FAILURE in 32m 53s","accounts_in_message":[],"_revision_number":2},{"id":"0ae339590d78a7654247e35fa50c9424190bef97","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-27 15:42:38.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"7772e5dfcf2fae39009e6337c002907a9c9e3045","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-27 15:49:08.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/55/722655/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b81a5ad : FAILURE in 28m 05s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/55/722655/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/a66c143 : FAILURE in 28m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"8cb5d677e76938cf9e3f916a19c8cef3a994a7a5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-04-27 15:53:13.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/55/722655/3/check/DellEMC_XtremIO_FC/e6165b5/DellEMC_XtremIO_FC/1185 : FAILURE in 32m 43s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/55/722655/3/check/DellEMC_XtremIO_iSCSI/7645ffb/DellEMC_XtremIO_iSCSI/1075 : FAILURE in 29m 43s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"bf9be30c3ff195fd7ce35a6923a3a749d64c5d74","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-27 16:00:25.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-3/ns5_nfs : FAILURE in 39m 41s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-3/ns5_iscsi : FAILURE in 39m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e25788c226cf513bd83183c120dd429b277edbfa","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-04-27 16:12:23.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9bbe10e6730227648e0312eb09882e1a313c2d2c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-27 16:12:59.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722655/3/iscsi/ : FAILURE in 21m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/722655/3/smb/ : FAILURE in 51m 46s","accounts_in_message":[],"_revision_number":3},{"id":"05e5b680212ed0799d5bb025ab01fe4431a16300","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-04-27 16:24:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/55/722655/3/check/fujitsu-eternus-dx-iscsi/8bfd609 : FAILURE in 43m 34s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/55/722655/3/check/fujitsu-eternus-dx-fc/d47cf87 : FAILURE in 1h 01m 50s","accounts_in_message":[],"_revision_number":3},{"id":"3f6f22522d461b8896e44ea704a2a81317111862","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-27 16:28:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2220f8a/2220f8a.html : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a4d272cde1de36fef84395405ff78410623548cb","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-04-27 16:40:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/55/722655/3/check/nimble-iscsi-driver-dsvm-volume/4e5f9fb : FAILURE in 48m 41s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/55/722655/3/check/nimble-fc-driver-dsvm-volume/a025044 : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3accd98dafe8fb66b50cc637d504d9c815d84965","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-27 17:18:42.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/55/722655/3/check-not-vote/ext-cinder-zuul/5237198 : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":3},{"id":"c130411f17c967f6e30de1ed6203099ef700a11b","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-27 17:56:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/856/index.html : FAILURE in 32m 23s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1005/index.html : FAILURE in 41m 04s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1044/index.html : FAILURE in 32m 58s","accounts_in_message":[],"_revision_number":3},{"id":"9a4c252c108617cca6e78bb550ebed4f6d341d53","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-27 18:37:54.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a63dd4288f73e28278405d63f5fcea20413aa292","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-27 18:51:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 32m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ac8f0e6e10a190e69451f76e4d21124ab330594f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-27 19:04:09.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/3/check/tempest-dsvm-full-bionic/7f6a9ae/ : SUCCESS in 1h 46m 29s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/3/check/devstack-plugin-ceph-dsvm-bionic/e52525a/ : FAILURE in 1h 19m 59s","accounts_in_message":[],"_revision_number":3},{"id":"602970dd82b3293402e7a62da488dfec8b0c0fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:06:58.000000000","message":"Patch Set 3: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a30f39e4b864c728478da89026d3385 : FAILURE in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9785861aa9e7470793412b9a490d757a : FAILURE in 12m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00b3c00e46ef427096bce1200c28408a : FAILURE in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb25b5b1c92e46f28a1364616577b473 : FAILURE in 12m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a41b8f5a94a4b219f45c58b2264fd51 : TIMED_OUT in 41m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0003bdc4e2b14e3eb31a23fdfb4e0df8 : SUCCESS in 22m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a49ea5a9194e423b8c0228c1aef1f83d : SUCCESS in 1h 19m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8fe2602285744358b5eb302c4b6c4d10 : FAILURE in 28m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d224044657bd4a2e8e2a53f8e2d743e5 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/767df77c34e24fbeab6d8cb3fe9dacef : FAILURE in 5m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ae6a2260973f46bc98b40681b2da246b : FAILURE in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aa90dee27ea7412fb92a9e22d614856a : FAILURE in 33m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/95cedcccb6394ae496a8822ddce01864 : FAILURE in 11m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be9abd7e8a47405d82b239e1498fe29d : SUCCESS in 1h 28m 57s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d7a243d153c240e1aa80c0f9bfc2dc56 : FAILURE in 33m 33s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ff77f718277b4171a67ed85944b747a8 : SUCCESS in 1h 26m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82df666b673d4d1fa06ef1f5b3aeee0b : FAILURE in 34m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2166ad041c9e47e3902b61066f939a69 : SUCCESS in 1h 57m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d54d5e371273428e8453938d06ac9d3a : FAILURE in 2h 29m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9361bf8eebf54abf82bc7c114c45f549 : FAILURE in 37m 07s","accounts_in_message":[],"_revision_number":3},{"id":"dcb339b76db54dc94c8a1f2754e611322f89a53f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-28 14:43:37.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/126/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"93fb616429b51823e7a9ac4dc377cb69845f71af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-04-28 14:44:02.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/125/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"fa10762de70aad1a8d9dccd827df5595c172a13b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 17:59:14.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"62a74c5c8271c400dfd6be6b3ea00d657c7e2d45","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 17:59:26.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"1e4026d974450f62fb4457f05d0b2ffb98f3ceff","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 17:59:41.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"6de24008400e8675ada88de6f0a086ea78ee933c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-06 11:55:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bcdd68a4ae175c44afac65f78afa2f4b92e00c63","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 12:10:47.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/0e508c42d57426ab7812616c209f0826998eeb4f/index.html : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"7acdfd85a3a1e8e257ce63456bc2599e6f79196b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-06 12:33:38.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-4/ns5_nfs : FAILURE in 37m 6s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-4/ns5_iscsi : FAILURE in 37m 20s","accounts_in_message":[],"_revision_number":4},{"id":"44dcfe8d794f9a996d4e7d355cd6ff90851fbf8e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-05-06 12:34:52.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/55/722655/4/check-not-vote/ext-cinder-zuul/5713ab6 : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":4},{"id":"44722ab00c154ce8798d2080dc977a12c4244765","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-05-06 12:45:36.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722655/4/iscsi/ : FAILURE in 49m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/722655/4/smb/ : FAILURE in 48m 51s","accounts_in_message":[],"_revision_number":4},{"id":"b5029fd8ef09d51e799dc5449e30878c93fffc95","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-05-06 12:47:40.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"c8536b11486093470367afbeb6855c4406964788","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-05-06 12:59:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/55/722655/4/check/fujitsu-eternus-dx-iscsi/86d2c0b : FAILURE in 35m 06s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/55/722655/4/check/fujitsu-eternus-dx-fc/32de64e : FAILURE in 34m 54s","accounts_in_message":[],"_revision_number":4},{"id":"7eb6b9c343b753e520d99eb864b82cf21f4de93d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-05-06 13:10:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/87c2197/87c2197.html : SUCCESS in 1h 13m 21s","accounts_in_message":[],"_revision_number":4},{"id":"547b659815637a3cd2a40d270b07a96132fb6499","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-05-06 13:30:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 32m 43s","accounts_in_message":[],"_revision_number":4},{"id":"0eab9ff51eea0cddbf3c812b04b577d7b31b2380","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 13:40:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/55/722655/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ffdf71b : FAILURE in 31m 33s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/55/722655/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f790125 : FAILURE in 1m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"8b57e4e97fc620e4caae5d3757fb31caf56a5a56","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-05-06 13:49:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/55/722655/4/check/cinder-storpool-tempest/977f4c3/ : RETRY_LIMIT in 13m 12s","accounts_in_message":[],"_revision_number":4},{"id":"9f262fc67d8bc9f9bdd4bb7793a7d4ee701792f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 14:14:46.000000000","message":"Patch Set 4: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec6021b6dcb3487cbdd2a55af7820d06 : FAILURE in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e9e05216944a67b089b30bed5be9d9 : FAILURE in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56de3a2437334c8ea2162ab30d106f9c : FAILURE in 11m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce1926a06f024127a335416d400643b8 : TIMED_OUT in 41m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/923315f6cf5b432188005332d2a983a7 : SUCCESS in 15m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/3ca145a9c32b4e29941a0d197241c7ea : SUCCESS in 1h 01m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7dbfe5d2d4524631b6d327581136fbe4 : FAILURE in 28m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5048ee41fcf4c70b099863a9c954bf1 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/916c7089b1b64a9ba2d32f67da4036c3 : FAILURE in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/14b65380fd05404c97510ec460c9af88 : FAILURE in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea172fcf405742b892e95530e1a7c7ed : FAILURE in 32m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21cbe2f703764cba98d9c290fd365673 : FAILURE in 6m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d106d45cffd547ea9eefeab3ac666da3 : SUCCESS in 1h 23m 54s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/38c26f14377f4200b9f796ed213d52da : FAILURE in 35m 20s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c7c12967e7e445da9686e0994656e43 : SUCCESS in 1h 30m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5734e8ffcb35438389644a4d9141da98 : FAILURE in 30m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/825f22bad1584cc6af8b8c0f5d55b2bb : SUCCESS in 1h 14m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c14b3a8cda704e588be092c812a685af : FAILURE in 2h 16m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/64e3968341234aa98ce160f71e81fcf3 : SUCCESS in 56m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faf970afb213404fb4fad52b9fcb3184 : FAILURE in 29m 10s","accounts_in_message":[],"_revision_number":4},{"id":"60571d5f9a6a38ceb6afc382c59bdecaca7cb12c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-06 14:21:14.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/4/check/tempest-dsvm-full-bionic/7a45c18/ : SUCCESS in 1h 48m 43s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/4/check/devstack-plugin-ceph-dsvm-bionic/d5fb3ea/ : FAILURE in 1h 08m 07s","accounts_in_message":[],"_revision_number":4},{"id":"ca4cdf012f32d951c24a460a59451b595f7f6c6d","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-05-06 15:21:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/55/722655/4/check/nimble-iscsi-driver-dsvm-volume/b406cc0 : UNSTABLE in 1h 23m 55s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/55/722655/4/check/nimble-fc-driver-dsvm-volume/65f0958 : UNSTABLE in 1h 18m 49s","accounts_in_message":[],"_revision_number":4},{"id":"6fc8d8301b0e68635afccc6e514032f4f15c7d2f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-06 16:27:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/55/722655/4/check/dsvm-tempest-infinibox-fc/844f02e : SUCCESS in 1h 09m 56s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/55/722655/4/check/dsvm-tempest-infinibox-iscsi/ba0aaa1 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":4},{"id":"92ff1d2bacd825117fdb219a8b3b58ae046790e8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 17:31:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/55/722655/4/check/DellEMC_XtremIO_FC/326d9ab/DellEMC_XtremIO_FC/1272 : FAILURE in 35m 25s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/55/722655/4/check/DellEMC_XtremIO_iSCSI/a772a2c/DellEMC_XtremIO_iSCSI/1159 : FAILURE in 30m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"b6ecabab72ad17631d45c0b00352adea7f60a644","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-06 18:37:13.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"4d537cc71f1e78993771d8c34bb71cbf6fd47e57","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-05-06 19:24:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/943/index.html : FAILURE in 32m 39s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1091/index.html : FAILURE in 41m 09s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1126/index.html : FAILURE in 32m 47s","accounts_in_message":[],"_revision_number":4},{"id":"06118670d49839176dc94dd2a0b32850c6ccd6ab","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-06 22:15:44.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e4a7aa613041651dc25f70b2d965cd80b55ce04e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-06 23:04:56.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/55/722655/4/check/DellEMC_Unity_FC/482c5bc/DellEMC_Unity_FC/1237 : FAILURE in 43m 51s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/55/722655/4/check/DellEMC_Unity_iSCSI/d608799/DellEMC_Unity_iSCSI/1109 : FAILURE in 29m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"06bff4de669e9b9c17832ad0aa11db515c64248a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-07 14:18:43.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"7fb7e5ec8fc8af941cd5674db0f320272ed5663a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-08-28 12:44:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b6d706d2a1e0ceef313a4cad43423e4471ca06bb","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-28 12:55:28.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/274996c3e0db0cf7400a67d0d1e7691ed76a7f21/index.html : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"bb21e8916c49c46d9cc4ef2bc992bd22daab9229","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-28 13:15:52.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-5/ns5_nfs : FAILURE in 27m 50s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-5/ns5_iscsi : FAILURE in 31m 16s","accounts_in_message":[],"_revision_number":5},{"id":"7b792aa2ebecb229da0cd04b3c1e9b2bfffda528","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-28 13:18:49.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"cf07b2b2f6291ac94a443cb82652c9056f0a81bc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-28 13:20:29.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/55/722655/5/check/DellEMC_XtremIO_FC/56e5ec6/DellEMC_XtremIO_FC/767 : FAILURE in 36m 00s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/55/722655/5/check/DellEMC_XtremIO_iSCSI/74a3d6c/DellEMC_XtremIO_iSCSI/733 : FAILURE in 28m 48s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"73ee31d5ad726dc226e0fbeff51cc2a534475d06","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-28 13:24:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/55/722655/5/check/dsvm-tempest-cisco-zonemanager-job/0f4d376/ : FAILURE in 32m 56s","accounts_in_message":[],"_revision_number":5},{"id":"41fb78b970e7f7721f9bd1af7f648d8a472b747d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-28 13:35:17.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"9160206087022f618e8afaee0ea9e4953b78788a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-28 13:52:05.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722655/5/iscsi/ : FAILURE in 48m 57s\n- cinder-smb http://cloudbase-ci.com/cinder/722655/5/smb/ : FAILURE in 1h 07m 19s","accounts_in_message":[],"_revision_number":5},{"id":"81f36df29cb0c65b8bb01dd448246e53934af7b3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-28 13:52:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/841b58a/841b58a.html : FAILURE in 1h 04m 57s","accounts_in_message":[],"_revision_number":5},{"id":"90220ec214124f47ef4df3777ee4182b3ba1e6b3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-28 14:33:09.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"acf83f31d2bb5ab98d646187fb5a5839034b6d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 15:09:53.000000000","message":"Patch Set 5: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a49678e81a0464cb50ad1c756491153 : FAILURE in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d92c327e0f42bca362fd514d2792df : FAILURE in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d48165e4f6947a8a00a6c444e70eee0 : FAILURE in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ead89d8c3aa40318ffa77a728ea150d : FAILURE in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2c0a85fffe4464b33f6e623b48d302 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/d2b8068844b64082bd910e19a9e56fca : SUCCESS in 1h 02m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/122c9f716b13485da1d2a5dbca548837 : FAILURE in 28m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c4b43bdb925f43f8afe1716316fc7db6 : FAILURE in 12m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0416f84897754d1c828986e1fcfad848 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/33d745009e0a44a9aa76562da4558ca8 : FAILURE in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77ab5d752fc744fca9ec6d7f6e233f0f : FAILURE in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9e9e79ecb19c4b75869810d4fcfe19ed : FAILURE in 36m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1b4f68d2d3c04a56a51ba3a5b9d9cbf5 : SUCCESS in 8m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d7381960b5874ce4b15cc90d63a02f4c : SUCCESS in 1h 05m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/657194ff851f415b9f7da70a5dd76174 : FAILURE in 1h 09m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cef07b29d6f844bbb362ba825ac47a38 : FAILURE in 31m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3d2da786b85b4d7aa4bb7c2ddf8edd84 : SUCCESS in 1h 09m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7a1c19ee0a644e1694417d76e75a1361 : FAILURE in 33m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6554eb978e574231832a6d13291340c1 : SUCCESS in 59m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/510b0581a74640278b684c3d6f4150a4 : FAILURE in 1h 26m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c9e41cc11624fc9a0c71e7e820fd155 : FAILURE in 37m 44s","accounts_in_message":[],"_revision_number":5},{"id":"444f8da9f630b55b201a8391c693ea8890b76bde","author":{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},"date":"2020-08-28 15:11:14.000000000","message":"Patch Set 5: Code-Review-1\n\nHello Ivan,\n\nPlease check Nexenta CI results (cinder backup log) - it failed to start\n\n2020-08-28 13:10:42.988 | + functions-common:service_check:1609      :   sudo systemctl status devstack@c-bak.service --no-pager\n2020-08-28 13:10:43.007 | ● devstack@c-bak.service - Devstack devstack@c-bak.service\n2020-08-28 13:10:43.007 |    Loaded: loaded (/etc/systemd/system/devstack@c-bak.service; enabled; vendor preset: enabled)\n2020-08-28 13:10:43.007 |    Active: failed (Result: exit-code) since Fri 2020-08-28 13:10:12 GMT; 30s ago\n2020-08-28 13:10:43.007 |   Process: 74490 ExecStart\u003d/usr/local/bin/cinder-backup --config-file /etc/cinder/cinder.conf (code\u003dexited, status\u003d1/FAILURE)\n2020-08-28 13:10:43.007 |  Main PID: 74490 (code\u003dexited, status\u003d1/FAILURE)\n2020-08-28 13:10:43.007 | \n2020-08-28 13:10:43.007 | Aug 28 13:10:11 openstack-master-ns5-nfs systemd[1]: Started Devstack devstack@c-bak.service.\n2020-08-28 13:10:43.007 | Aug 28 13:10:12 openstack-master-ns5-nfs cinder-backup[74490]: INFO cinder.cmd.backup [-] Backup running with 1 processes.\n2020-08-28 13:10:43.007 | Aug 28 13:10:12 openstack-master-ns5-nfs cinder-backup[74490]: DEBUG oslo_concurrency.lockutils [-] Acquired lock \"singleton_lock\" {{lock /usr/local/lib/python3.7/dist-packages/oslo_concurrency/lockutils.py:266}}\n2020-08-28 13:10:43.007 | Aug 28 13:10:12 openstack-master-ns5-nfs cinder-backup[74490]: DEBUG oslo_concurrency.lockutils [-] Releasing lock \"singleton_lock\" {{lock /usr/local/lib/python3.7/dist-packages/oslo_concurrency/lockutils.py:282}}\n2020-08-28 13:10:43.007 | Aug 28 13:10:12 openstack-master-ns5-nfs cinder-backup[74490]: CRITICAL cinder [-] Unhandled error: TypeError: \u0027NoneType\u0027 object is not iterable\n2020-08-28 13:10:43.008 |                                                                ERROR cinder Traceback (most recent call last):\n2020-08-28 13:10:43.008 |                                                                ERROR cinder   File \"/usr/local/bin/cinder-backup\", line 10, in \u003cmodule\u003e\n2020-08-28 13:10:43.008 |                                                                ERROR cinder     sys.exit(main())\n2020-08-28 13:10:43.008 |                                                                ERROR cinder   File \"/opt/stack/cinder/cinder/cmd/backup.py\", line 123, in main\n2020-08-28 13:10:43.008 |                                                                ERROR cinder     for backend in filter(None, CONF.enabled_backup_backends):\n2020-08-28 13:10:43.008 |                                                                ERROR cinder TypeError: \u0027NoneType\u0027 object is not iterable\n2020-08-28 13:10:43.008 |                                                                ERROR cinder \n2020-08-28 13:10:43.008 | Aug 28 13:10:12 openstack-master-ns5-nfs systemd[1]: devstack@c-bak.service: Main process exited, code\u003dexited, status\u003d1/FAILURE\n2020-08-28 13:10:43.008 | Aug 28 13:10:12 openstack-master-ns5-nfs systemd[1]: devstack@c-bak.service: Failed with result \u0027exit-code\u0027.","accounts_in_message":[],"_revision_number":5},{"id":"1679ff30251d6b877d5b6ca2a770a274ca6f6550","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-28 18:06:29.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/5/check/tempest-dsvm-full-bionic/274a8af/ : SUCCESS in 1h 37m 13s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/5/check/devstack-plugin-ceph-dsvm-bionic/7f95f40/ : FAILURE in 56m 55s","accounts_in_message":[],"_revision_number":5},{"id":"cc0ea9ee51e6631bcc9eff0b39c64c55bc1d4256","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-28 19:26:37.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"e341f65aad9ad1e0091ae8ca048a9e951457242d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-29 04:21:45.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/55/722655/5/check/DellEMC_Unity_FC/7a83a3b/DellEMC_Unity_FC/530 : FAILURE in 40m 51s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/55/722655/5/check/DellEMC_Unity_iSCSI/fd5be28/DellEMC_Unity_iSCSI/498 : FAILURE in 28m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"0474b12ee0da44af4fc4d7a7133b64803ae239a6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-29 04:56:34.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"40633198ff963a358522ce233582e97f23222888","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-29 06:31:03.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/55/722655/5/check/nimble-iscsi-driver-dsvm-volume/dbf32f8 : FAILURE in 21s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/55/722655/5/check/nimble-fc-driver-dsvm-volume/7ee5377 : FAILURE in 5s","accounts_in_message":[],"_revision_number":5},{"id":"3f150c6d4b306aa7062baeea584ac564953058eb","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-29 11:53:12.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1614/index.html : FAILURE in 33m 38s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1758/index.html : FAILURE in 18s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1772/index.html : FAILURE in 33m 34s","accounts_in_message":[],"_revision_number":5},{"id":"f78f7b0273ee0e11f47974050b0bf46c68241622","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-29 12:18:30.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/55/722655/5/check-not-vote/ext-cinder-zuul/753e0cc : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":5},{"id":"d8de97af51085715bc9f8a28db77fb6f02d8b03e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-29 12:26:50.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"ac0ad0831a34536be9aca11a6c5f444eb3ac4a35","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-08-30 12:41:41.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/55/722655/5/check/DellEMC_PowerMAX_FC/07a2f3d/DellEMC_PowerMAX_FC/555 : FAILURE in 36m 10s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/55/722655/5/check/DellEMC_PowerMAX_iSCSI/ae1a7ec/DellEMC_PowerMAX_iSCSI/482 : FAILURE in 5m 37s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"a73553874e3d13ad8981a865e2f1754f906d08d3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-30 13:14:35.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"909b8f8cfa5c75f1c907c9a7be1c685439684f14","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-30 21:17:21.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1630/index.html : FAILURE in 15m 57s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1774/index.html : FAILURE in 45s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1788/index.html : FAILURE in 28m 32s","accounts_in_message":[],"_revision_number":5},{"id":"696e938fcaf73c9bb934c32040bdccac463c6b5f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-30 21:20:05.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c83af46142b0c4c15b658f853263261b45eb6f02","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-01 07:42:12.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/5/20200901160959/NEC-ISCSI : FAILURE in 30m 47s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/5/20200901160959/NEC-FC : FAILURE in 32m 11s","accounts_in_message":[],"_revision_number":5},{"id":"0fe6c7d6ad92695518b5c461dc6e8507f107e266","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-01 08:24:28.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"b359de56965d85cb81ad9b62d6b545cc80363074","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-01 18:33:26.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-5/ns5_nfs : FAILURE in 2h 33m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-5/ns5_iscsi : FAILURE in 3h 30m 46s","accounts_in_message":[],"_revision_number":5},{"id":"bac60775202aae034a2c950c4c51a464d31a8e3a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-11 15:53:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b3fab643cb7b159f5aa14879b1a727548f06fac1","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-11 15:56:16.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/55/722655/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/20cefd4 : FAILURE in 2m 33s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/55/722655/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/dfc500f : FAILURE in 2m 20s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"8eb95caba656b1c578e82977bae1f68cb4cb3be0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-11 16:19:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/bb73ee6/bb73ee6.html : POST_FAILURE in 7m 23s","accounts_in_message":[],"_revision_number":6},{"id":"e27cc642f8e9f1d37f63a21d92b8fd42f4397920","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-11 16:26:38.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722655/6/iscsi/ : FAILURE in 32m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/722655/6/smb/ : FAILURE in 32m 41s","accounts_in_message":[],"_revision_number":6},{"id":"3e99fd3938923f5ee61ffff230db280bfbaa80be","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-11 16:58:05.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"bf156a8ca118266ea9d6b50293a0a7d4babeff43","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-11 17:59:45.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-6/ns5_nfs : SUCCESS in 2h 4m 9s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-6/ns5_iscsi : SUCCESS in 2h 5m 58s","accounts_in_message":[],"_revision_number":6},{"id":"0314d5c0eeacfb8fde38774e69cd51dfd0505851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:06:34.000000000","message":"Patch Set 6: Verified-1\n\n(16 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e18a7a630f14b2997d2cc2a10b9be13 : FAILURE in 10m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d74907526234b88ac13225ea050a18d : TIMED_OUT in 41m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7702d69493674f81abc2bba81fb4d4e8 : FAILURE in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52441b711d1445e4995a5e0ccf09dfbd : SUCCESS in 22m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/783f16e921024eff9c49820dc5e4bf27 : SUCCESS in 1h 01m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17bb8d494b554895951688894cc26504 : SUCCESS in 1h 11m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c3f33bf878974502bd6a95fbb07ac61d : TIMED_OUT in 41m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40d5ab44b32c4482956e6a253d8f5640 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/152f3ec262874c7c8fca57765c237cf4 : FAILURE in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1365170a9aa64cf9b1dc824f387ba6e2 : FAILURE in 14m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8aae6808d0aa43f5858b2216d76dfc14 : FAILURE in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e1c4104216bc48988ee785c792ef1422 : SUCCESS in 39m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/339b9155a0b347feb0a6c6cb2e1e0186 : SUCCESS in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/02d69ff7c7d34f61b3e2c5db28546f8d : SUCCESS in 1h 26m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/664aea2eea7b46cab015ec9fb1df743e : FAILURE in 1h 26m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ff2c5eefa4a44d00819baa3f416dbfd5 : SUCCESS in 1h 00m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8a83f81858e466ebee42db05d24376d : SUCCESS in 55m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09ef11c62e154e618b0a01c8e994a8c7 : SUCCESS in 1h 21m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/94fd1ec3cfa548eebb93dbed9d797f30 : FAILURE in 1h 03m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1ae8c5ee0834640bb1a25c471c75005 : SUCCESS in 2h 00m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8670c4eeee84d0ea505ae24b2024da1 : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":6},{"id":"1e07c86ff3b35d82e5ea80c6cf5f89aa585b1299","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-11 20:05:25.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"1ef09ccc9991b040ca209e4d7a3bd7e27ffdac1e","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-11 20:09:11.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/7e472abe13b6fbcab4e87434edf1345a1ff561c9/index.html : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"f2ce9d769968bf754712ce71ba68b672326e9d2a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-11 20:39:34.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"034ec3e29fcb3988dd221ca77fc765359731a4b4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-12 01:31:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/55/722655/6/check/cinder-storpool-tempest/c2cbf24/ : SUCCESS in 1h 09m 03s","accounts_in_message":[],"_revision_number":6},{"id":"3db48e1c2b8377aac13bad5127ac779cd99ea316","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-12 02:01:52.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/55/722655/6/check/DellEMC_XtremIO_FC/3c0da1d/DellEMC_XtremIO_FC/871 : SUCCESS in 1h 34m 49s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/55/722655/6/check/DellEMC_XtremIO_iSCSI/0e498b0/DellEMC_XtremIO_iSCSI/835 : SUCCESS in 1h 39m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"facd33698ca18b295b871df7d3a259d762be7a5f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-12 02:25:01.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/6/check/tempest-dsvm-full-bionic/04d472f/ : SUCCESS in 1h 34m 53s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/6/check/devstack-plugin-ceph-dsvm-bionic/2e303cc/ : SUCCESS in 1h 37m 49s","accounts_in_message":[],"_revision_number":6},{"id":"5b3c8ad38c4183882ced60f6ea7560e413f70c69","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-12 02:47:04.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"56ca5959cf1baf3f9df2de18c44025418195015f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-12 07:02:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/6/20200912134148/NEC-ISCSI : SUCCESS in 2h 17m 32s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/6/20200912134148/NEC-FC : SUCCESS in 2h 20m 19s","accounts_in_message":[],"_revision_number":6},{"id":"e214d23763367d14fbc8756b4697e68add472354","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-12 07:51:24.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/55/722655/6/check/DellEMC_PowerMAX_FC/c0ac31d/DellEMC_PowerMAX_FC/641 : SUCCESS in 3h 52m 22s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/55/722655/6/check/DellEMC_PowerMAX_iSCSI/8233cdb/DellEMC_PowerMAX_iSCSI/562 : SUCCESS in 4h 32m 04s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"04a4eb5d2d30b068f9ef48f34885a0b312f83d38","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-12 08:26:10.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"3045227f19609206482321443c49bf3ad5811972","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-12 08:26:46.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/55/722655/6/check/dsvm-tempest-zadara-driver/2d4a6d5 : FAILURE in 1h 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"08005bbe0cfc797a9fde8119390d71b403cc7d56","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-18 10:30:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"22e5aa10d8806d4ace5ff05721b1ae9aaef83a81","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-18 10:31:09.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/55/722655/7/check/fujitsu-eternus-dx-iscsi/c5857e9 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/55/722655/7/check/fujitsu-eternus-dx-fc/f73ac3a : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"5214ff302271dc4ff8bcc5250d656d46fb6f4deb","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-18 10:42:30.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/cb1dc3c0814cbedc0887d7f7518571eadccdec2f/index.html : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"167363076947f8d4cca514c039f1e4b63fe221c1","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-18 10:53:26.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/55/722655/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c04e730 : FAILURE in 22m 23s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/55/722655/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/78c5f8b : FAILURE in 21m 43s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"10a4a868a642f08a2f91bfa28189268a252dd939","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-18 11:14:13.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-7/ns5_nfs : FAILURE in 43m 6s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-7/ns5_iscsi : FAILURE in 42m 28s","accounts_in_message":[],"_revision_number":7},{"id":"39ebee75fbf506ebab76cd2e540dfb741e7b4b0e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-18 11:21:45.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"dd69d3552a69e208a7ab169a3da57daf860d50c5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-18 11:30:28.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722655/7/iscsi/ : SUCCESS in 59m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/722655/7/smb/ : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":7},{"id":"4fb5c5eac2ccdc52e1ba0ebc1bf953aa6c671bf8","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-18 11:35:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"546aa392a95164b9ba0ba3749aafceea66bdb0f0","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-18 11:35:38.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/55/722655/8/check/fujitsu-eternus-dx-iscsi/d1d5c0c : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/55/722655/8/check/fujitsu-eternus-dx-fc/b616e42 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"b7d4d54916446e0442d03d5820b554a848f8d306","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-18 11:47:25.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/b2ba48aff09216545c4b35153092222e427c2fca/index.html : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"3686b92195bd67b31ca754ffb8a40d72bbfc9447","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-18 11:58:52.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/55/722655/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/af0dc0c : FAILURE in 22m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/55/722655/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/45cbff1 : FAILURE in 21m 39s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"f0bfe10fea0c512e3e583167d582b0c4650f0ae8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-18 12:19:00.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-8/ns5_nfs : FAILURE in 43m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-8/ns5_iscsi : FAILURE in 43m 31s","accounts_in_message":[],"_revision_number":8},{"id":"84d158ac19d6b446df82f619a4856e01b837eb6d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-18 12:26:05.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"5f38cd555a3ba0d4604a7383860c66a47d771f58","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-18 12:38:26.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722655/8/iscsi/ : SUCCESS in 1h 02m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/722655/8/smb/ : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":8},{"id":"096d9889bbe1bf7968d37583f76248d7618d8a28","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-18 12:47:44.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/7bda6d0/7bda6d0.html : SUCCESS in 1h 09m 11s","accounts_in_message":[],"_revision_number":8},{"id":"927b87be13f08d7944fb9810924f61f2690921a4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-18 13:49:26.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"d2e768c6903438cd828575ec161ed3aeefb05fb5","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-18 13:49:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dfeffb9c30d748855265e44063a1b9423e952e12","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-18 13:50:23.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/55/722655/9/check/fujitsu-eternus-dx-iscsi/39c48dc : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/55/722655/9/check/fujitsu-eternus-dx-fc/00ef670 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"ffd664136275295c93c8a16b2eee7d2318f7aaaa","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-18 13:54:06.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"e6a889b35083fa8503917b3a6f5533e7f1064baf","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-18 14:02:52.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/9d8ae4c92d3762238bc95e17e0f365b5eda4676c/index.html : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"8543bc7e8d6c81a8ad231b1553ee205194da89d0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-18 14:13:18.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/55/722655/9/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/d442995 : FAILURE in 22m 19s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/55/722655/9/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/5a6f3d6 : FAILURE in 21m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"a9a8be2ac6a01c019df36d66f1cdb502a33a5214","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-18 14:32:40.000000000","message":"Patch Set 9:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-55-722655-9/ns5_nfs : FAILURE in 42m 13s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-55-722655-9/ns5_iscsi : FAILURE in 42m 13s","accounts_in_message":[],"_revision_number":9},{"id":"f7376567b6a2d4dde4326f7a28edd690200f8a4b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-18 14:47:46.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/722655/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"41c3e9cbc09b84f6c55935f78c8b4004c57a6b39","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-18 15:03:20.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ae31efd/ae31efd.html : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":9},{"id":"7054a9751a52fd6f203627841f78b890bd3f5521","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-18 15:08:33.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/722655/9/iscsi/ : SUCCESS in 1h 18m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/722655/9/smb/ : FAILURE in 32m 09s","accounts_in_message":[],"_revision_number":9},{"id":"5f4f4df7fb880c193cb801a13c940f57f6546255","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-18 16:11:28.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"3b8a5e14735ea79b1605b56f83c226bf678c2e4b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-18 16:24:58.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"ffa91607e892a2a1f29414eae149acd0ec984dd8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-18 18:03:19.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"5d8946a4cb0891aec8440d3f4fa0d5921fbe7ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 18:07:58.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7dac9df8afdf43bb98f6f92fd0776a4d : FAILURE in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36688f5319e46f39666304801dba4ff : FAILURE in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5380fd844c2d4a1c9bec10349408594d : TIMED_OUT in 41m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb31b5914cd462ab6882a5d2b9a982f : FAILURE in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463d6ee2a9dd4805b5e375307b2fced6 : SUCCESS in 21m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/fbd0eaf0657947a0bac56e5d05cf8881 : SUCCESS in 1h 04m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0aaf3776b0a043d9bc7bec8b5cff11b6 : SUCCESS in 1h 02m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fc66051a04f24ca3a8286de689a9b43d : FAILURE in 11m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d61777817fba4c18b12f1bb0cf3ff73d : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e46f0f4d0f5640eeba955517daa78319 : FAILURE in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/148922d4cb4d4ed6b6df0763c99768a8 : FAILURE in 4m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/659c556fb5a442d88523a714011e5606 : SUCCESS in 43m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6fddcb7489a14133b4a093f16076ab51 : FAILURE in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e507e629e4144a8a8f496ad12303539 : SUCCESS in 1h 00m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/36b3db21ab024ff6a093df3d3ba50d09 : FAILURE in 1h 08m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dad94764b1c84a7ca7be6b7dccf5dd58 : SUCCESS in 1h 39m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f1994773d64c44ac99b37083c7dbc81f : SUCCESS in 1h 15m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2fb706f58cb0448d97a9117d4e03b33c : SUCCESS in 1h 06m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a61acbf8baf41118847754c06da2715 : SUCCESS in 1h 11m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab8c32473af64e11b3540a2792ac4775 : SUCCESS in 2h 00m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/420d367c5b7a439eb3773a2e0a8130e1 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":9},{"id":"b7b264ed0126011f4d77c736f8d1b7785c1ab1ae","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-18 18:38:02.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"c58dd9d973d5ab5fac29e38bcda3c9683bba13a0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-18 18:57:21.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/55/722655/9/check/DellEMC_XtremIO_FC/8655c29/DellEMC_XtremIO_FC/933 : SUCCESS in 1h 34m 22s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/55/722655/9/check/DellEMC_XtremIO_iSCSI/1586b73/DellEMC_XtremIO_iSCSI/895 : SUCCESS in 1h 37m 19s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"0245e1e8cdb9fdedc6dcafc70736c396ef439439","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-18 19:15:40.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/55/722655/9/check/DellEMC_Unity_FC/8159dab/DellEMC_Unity_FC/673 : SUCCESS in 1h 13m 38s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/55/722655/9/check/DellEMC_Unity_iSCSI/c32f592/DellEMC_Unity_iSCSI/641 : SUCCESS in 1h 49m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"b4a7121e3ff746965f2521dcfbab2726b4ba8a03","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-18 19:39:18.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/9/check/tempest-dsvm-full-bionic/db2c9bb/ : SUCCESS in 1h 23m 57s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/9/check/devstack-plugin-ceph-dsvm-bionic/24c1875/ : SUCCESS in 1h 32m 46s","accounts_in_message":[],"_revision_number":9},{"id":"2cd5b35bf121787b45ce6052ab3a2cb58ba40e19","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-18 20:57:11.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/55/722655/9/check/cinder-storpool-tempest/3ad2ada/ : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":9},{"id":"8364147e2355c6dbfbafdac2d8b1f27b181192da","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-18 22:13:53.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"12023c84fc27b6363d26d9fd8f866a35c8cebb76","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-19 23:09:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/9/20200920054204/NEC-FC : SUCCESS in 2h 21m 57s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/9/20200920054204/NEC-ISCSI : SUCCESS in 2h 27m 43s","accounts_in_message":[],"_revision_number":9},{"id":"06202f42eb44b8be8e4d8b3da47f371bbbe62c8c","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-09-21 03:28:25.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/55/722655/9/check/nimble-iscsi-driver-dsvm-volume/7f9dd0f : SUCCESS in 1h 35m 29s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/55/722655/9/check/nimble-fc-driver-dsvm-volume/b06be9f : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":9},{"id":"fc3273573c0a537252c908ffba4bce5422dd60b4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-25 15:34:37.000000000","message":"Patch Set 9:\n\n* NexentaStor5_Cinder_NFS http://deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-55-722655-9/ns5_nfs : FAILURE in 2h 59m 0s\n* NexentaStor5_Cinder_iSCSI http://deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-55-722655-9/ns5_iscsi : SUCCESS in 2h 52m 58s","accounts_in_message":[],"_revision_number":9},{"id":"2518dbec76289c826cc337da76b5c0ff6d9a7274","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-25 18:22:09.000000000","message":"Patch Set 9:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-55-722655-9/ns5_nfs : SUCCESS in 2h 43m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-55-722655-9/ns5_iscsi : SUCCESS in 2h 46m 44s","accounts_in_message":[],"_revision_number":9},{"id":"3e7f147427e78ddb976c1b436a182c48deff5baf","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-10-02 10:16:09.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1c4d73090cba628f7aaaa1e63f1a41a489586af0","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-10-02 10:16:38.000000000","message":"Patch Set 10:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/55/722655/10/check/fujitsu-eternus-dx-iscsi/532304f : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/55/722655/10/check/fujitsu-eternus-dx-fc/15f41a4 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":10},{"id":"bbc8f20958bd192587e0d7656ea124b4ac782f9c","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-02 10:33:03.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/8b0a12d71f412fb81ade51a8a53fc5f9f0fa547e/index.html : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"08a3e6063214a12a119f8c72a9c73336aec4a0c0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-02 10:58:41.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/55/722655/10/check-not-vote/ext-cinder-zuul/3e32b94 : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":10},{"id":"169ee6be3cd5e15bf309b979e50015e3c4dc8f94","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-02 11:17:36.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/10/check/tempest-dsvm-full-bionic/6059c90/ : FAILURE in 55m 29s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/10/check/devstack-plugin-ceph-dsvm-bionic/d73ef49/ : FAILURE in 21m 14s","accounts_in_message":[],"_revision_number":10},{"id":"d1286e7e1f368df64ccf69a9b8d823687a1729bb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-10-02 11:27:21.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/5cd7d67/5cd7d67.html : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":10},{"id":"2de8054e05496a25f34dc5db49118376390b7537","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-02 11:40:45.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"1ecc4661411098daafadbfbce5054396b1620e04","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-02 11:45:35.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"5d7b78f7f7a264014a436343db65998119092f6a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-10-02 15:15:54.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a4574ca3d75e86ffbb78102a1864cea8800f4312","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-10-02 15:16:18.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/55/722655/11/check/fujitsu-eternus-dx-iscsi/59ab770 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/55/722655/11/check/fujitsu-eternus-dx-fc/272c78b : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"0e85bad4f0a95b59142f209eb31f6fb0112630f4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-02 15:22:31.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/55/722655/11/check-not-vote/ext-cinder-zuul/083189f : FAILURE in 6m 00s","accounts_in_message":[],"_revision_number":11},{"id":"37435996efd2941bc1abc2eee1b2d5f6fa2293ff","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-02 15:33:03.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/c2e173ec7109ddcc405aba7bb13d015a31fdd603/index.html : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"b769b875e12ddf2ed6dcc6909d39f9625e06477e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-02 15:49:32.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"38acf6cd90dee50c66ac669f493dc99edb5d7cbd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-02 15:50:01.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"c36a2d93f3ad274b642b06f3a578412bdb8a7d9e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-10-02 16:25:46.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/10/20201002224330/NEC-FC : SUCCESS in 2h 18m 12s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/10/20201002224330/NEC-ISCSI : SUCCESS in 2h 42m 14s","accounts_in_message":[],"_revision_number":10},{"id":"c39027e2cccb1ee1c6a71df1a514afa12a088902","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-10-02 16:28:28.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/55/722655/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a342c24/a342c24.html : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":11},{"id":"17df31e4652e83946c8667b9acaad815f5898416","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-02 16:41:05.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/11/check/tempest-dsvm-full-bionic/08feace/ : FAILURE in 1h 15m 39s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/11/check/devstack-plugin-ceph-dsvm-bionic/81bbc1b/ : FAILURE in 25m 38s","accounts_in_message":[],"_revision_number":11},{"id":"ec7e0984c2f0ced06c2f4e0956f5432d5f9040bc","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-02 17:17:25.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"f0feea0a6d1bce058f8d5d8432d2a827b8599101","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-02 17:25:40.000000000","message":"Patch Set 10:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-55-722655-10/ns5_nfs : SUCCESS in 4h 28m 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-55-722655-10/ns5_iscsi : SUCCESS in 4h 4m 59s","accounts_in_message":[],"_revision_number":10},{"id":"2272962e7400e4241e3dfe4f1bf2967faafcf213","author":{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"},"date":"2020-10-02 17:32:25.000000000","message":"Patch Set 11: Code-Review+1\n\nThank you!","accounts_in_message":[],"_revision_number":11},{"id":"6d8da094334a7e838046ad966d33aec3bf29e3b2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-02 20:19:25.000000000","message":"Patch Set 11:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-55-722655-11/ns5_nfs : SUCCESS in 3h 51m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-55-722655-11/ns5_iscsi : SUCCESS in 3h 46m 11s","accounts_in_message":[],"_revision_number":11},{"id":"faa2e37bbc0a6b3ddff0a9f976369a99f81adeb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 21:02:16.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6116b8a2bb824b2d97e4089cc7fb8781 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb331539855048b8a48cb8e3c13bff84 : FAILURE in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e631c2384a0461395b39e99f4c68a28 : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a57f7ce2c4d4163832def8ac9c6f4ae : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0dc9fef6ad2412596775843c39a95a4 : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/7dc1cdb797094a95b1dd511c37ddfab4 : SUCCESS in 1h 23m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5e9dd73233f40ca9c14986e58a6d3b9 : SUCCESS in 1h 15m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/675ed5705dad46f58f61600cac893da2 : SUCCESS in 12m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/53d9602db11c4a94bb83cf068a3545d3 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2448ba27c6be4b7c97b9e7ad38cf30ce : FAILURE in 6m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/281e9ceab1da421f8942d1fe436b5331 : FAILURE in 5m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2048eb9ff7f0437791225a6c9391fa51 : SUCCESS in 43m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/335221af21974833bc7ec1cd86c1c754 : FAILURE in 4m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7968a8829b4d4092a3804b6637e89182 : SUCCESS in 1h 04m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4b7604d133d64b3cae581044a4b398f1 : FAILURE in 1h 18m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a79edd20319442959d0aad32d6a7cb64 : FAILURE in 1h 27m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c5f8a915240f4b598d60a4197aa5d26d : SUCCESS in 1h 11m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/793f17681d844e1f9dee02c77cc94945 : SUCCESS in 1h 32m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/53139818622d4b90ac0a93dbfbffe033 : SUCCESS in 1h 17m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a6bcb509e41440f186b3e20059ce16e4 : FAILURE in 1h 58m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e173d2051bb42118893de110c2fc24f : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":11},{"id":"b8cb1b9fa07cecd09cf24f4f219842cdb12a783a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-10-02 21:20:13.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/11/20201003035217/NEC-FC : SUCCESS in 2h 26m 46s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/55/722655/11/20201003035217/NEC-ISCSI : SUCCESS in 2h 27m 53s","accounts_in_message":[],"_revision_number":11},{"id":"758e48675ced3dc3f2127d431a4701810df7b3d1","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-10-03 07:21:37.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/55/722655/11/check/dsvm-tempest-zadara-driver/ce1d887 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4f3a30b791674dd3baafa2c9b9340c68402a595a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-10-04 01:06:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/55/722655/11/check/cinder-storpool-tempest/967182c/ : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":11},{"id":"2207859bf9be6d98b965908eb80985ff6a311539","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-10-04 13:25:42.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/55/722655/11/check/nimble-iscsi-driver-dsvm-volume/58c6edb : SUCCESS in 3h 56m 08s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/55/722655/11/check/nimble-fc-driver-dsvm-volume/03bb46a : SUCCESS in 3h 03m 50s","accounts_in_message":[],"_revision_number":11},{"id":"57d01f6d24e662e7e05a60604811e7d11460b6bd","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-10-12 18:29:26.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"de85b16490fddbe00b1371d8104a6984be55e62c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-12 19:05:24.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"0f14154d9f6cbd7b85bd9c41e37f46d40127335c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-10-12 19:23:43.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nPossible problem noted inline.","accounts_in_message":[],"_revision_number":11},{"id":"599659abeba8b9869f99fad4d1f6c50a302f5eb8","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-10-12 19:35:42.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"93761f0aa7595d721d0f2222ae0e00ae5ea3936d","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-10-12 19:36:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"56effa6e5502cbfb6eaf00187810d11cd0832cc4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-12 22:03:49.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/11/check/tempest-dsvm-full-bionic/635cc05/ : FAILURE in 1h 59m 16s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/55/722655/11/check/devstack-plugin-ceph-dsvm-bionic/bd704cf/ : FAILURE in 28m 52s","accounts_in_message":[],"_revision_number":11},{"id":"edea3ddcf49f0197d24816d92056c320e12e11bd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-12 22:40:28.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/55/722655/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"75f41d8f86e3ed284545c92cb835d5af1f7b5df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 05:34:53.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/520b4678459a4ef69beacdbe35b29f34 : SUCCESS in 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ff2501a40e4e629f3c3c8a6dcb4acc : FAILURE in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742f2a130d084ee195634a43f6719219 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4472bac6573344e78aba1d952f5abe3f : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d4b8775df541fb8d7b789bcbf55db1 : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/908c6b61e668465fad64f8b5848ea3ce : SUCCESS in 58m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7309b8e4393647b69417c3e5c0ede4b5 : FAILURE in 58m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/985adb05534f4c9baf4edcff06942675 : SUCCESS in 12m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e5d7e7408464d3ca59c7fa848b59bed : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94a0dd76ea654223ab1023a2f20e99f6 : FAILURE in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f48ffb21ac2349b7afbc71323fe5a0ce : FAILURE in 4m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b0a537325d084478af161f5b201a91c9 : SUCCESS in 51m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8def6ef52754834a2c8e8f2671ebeaf : FAILURE in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7c0677038ca6447191530c5170179ef5 : SUCCESS in 1h 04m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d433aaecdeae422d97a892b21f6363f4 : FAILURE in 1h 08m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b179a8cd5eaf45da83ed10c75656af06 : SUCCESS in 1h 17m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eaa72448d67e4df7bf93045a97680b40 : SUCCESS in 1h 05m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d84ebbfab24a4ba797c2534ed8113300 : SUCCESS in 1h 13m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/779448cbb57a45e99f6335e1a0d88130 : SUCCESS in 1h 21m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0236a6005d2141848ae5c2508f741864 : FAILURE in 2h 22m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5cdd5a80e7c4f0ca84546d6f854453f : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":11},{"id":"eadd54a849f6ca116c390805d3b715ee42441049","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-10-14 08:47:56.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/55/722655/11/check/cinder-storpool-tempest/c8aafd3/ : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":11},{"id":"19f7a9fb8534439a62e8eae1caeceb99c7be8ad9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-14 12:31:21.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/55/722655/11/check-not-vote/ext-cinder-zuul/caeea76 : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":11},{"id":"a95467a087716342be9680f4ba81a15a9a39d4b6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-27 21:13:31.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/55/722655/11/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/cf90f4c : FAILURE in 21m 03s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/55/722655/11/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/94fe072 : FAILURE in 33m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"c2e173ec7109ddcc405aba7bb13d015a31fdd603","revisions":{"5ae54b3900cd319f139ac77ec0dd28c02e4e9baf":{"kind":"REWORK","_number":1,"created":"2020-04-24 10:26:03.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/1"}}},"commit":{"parents":[{"commit":"595b8adc4415abb0b6a67c9910890942b82a9e94","subject":"Merge \"Revert \"Mark MacroSAN Driver Unsupported\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/595b8adc4415abb0b6a67c9910890942b82a9e94"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-24 10:25:27.000000000","tz":180},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ae54b3900cd319f139ac77ec0dd28c02e4e9baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ae54b3900cd319f139ac77ec0dd28c02e4e9baf"}]},"branch":"refs/heads/master"},"fb4d9f0ffb937ca7835e2ef8982600db8565368d":{"kind":"REWORK","_number":2,"created":"2020-04-27 14:50:26.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/2"}}},"commit":{"parents":[{"commit":"595b8adc4415abb0b6a67c9910890942b82a9e94","subject":"Merge \"Revert \"Mark MacroSAN Driver Unsupported\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/595b8adc4415abb0b6a67c9910890942b82a9e94"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-27 14:50:08.000000000","tz":180},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb4d9f0ffb937ca7835e2ef8982600db8565368d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb4d9f0ffb937ca7835e2ef8982600db8565368d"}]},"branch":"refs/heads/master"},"3aabc325b04ffe3f0520b8a657adbb4d5e7c23e5":{"kind":"REWORK","_number":3,"created":"2020-04-27 15:20:07.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/3"}}},"commit":{"parents":[{"commit":"595b8adc4415abb0b6a67c9910890942b82a9e94","subject":"Merge \"Revert \"Mark MacroSAN Driver Unsupported\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/595b8adc4415abb0b6a67c9910890942b82a9e94"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-27 15:20:04.000000000","tz":0},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3aabc325b04ffe3f0520b8a657adbb4d5e7c23e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3aabc325b04ffe3f0520b8a657adbb4d5e7c23e5"}]},"branch":"refs/heads/master"},"0e508c42d57426ab7812616c209f0826998eeb4f":{"kind":"REWORK","_number":4,"created":"2020-05-06 11:55:54.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/4"}}},"commit":{"parents":[{"commit":"e82ff57b70bb94c39c0eacf9d32a38a411571581","subject":"Merge \"Monkey patch original current_thread _active\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e82ff57b70bb94c39c0eacf9d32a38a411571581"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-06 11:55:32.000000000","tz":180},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e508c42d57426ab7812616c209f0826998eeb4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e508c42d57426ab7812616c209f0826998eeb4f"}]},"branch":"refs/heads/master"},"274996c3e0db0cf7400a67d0d1e7691ed76a7f21":{"kind":"REWORK","_number":5,"created":"2020-08-28 12:44:12.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/5"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-08-28 12:43:56.000000000","tz":180},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/274996c3e0db0cf7400a67d0d1e7691ed76a7f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/274996c3e0db0cf7400a67d0d1e7691ed76a7f21"}]},"branch":"refs/heads/master"},"7e472abe13b6fbcab4e87434edf1345a1ff561c9":{"kind":"REWORK","_number":6,"created":"2020-09-11 15:53:20.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/6"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-11 15:53:09.000000000","tz":0},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e472abe13b6fbcab4e87434edf1345a1ff561c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e472abe13b6fbcab4e87434edf1345a1ff561c9"}]},"branch":"refs/heads/master"},"cb1dc3c0814cbedc0887d7f7518571eadccdec2f":{"kind":"REWORK","_number":7,"created":"2020-09-18 10:30:40.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/7"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-18 10:30:20.000000000","tz":0},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb1dc3c0814cbedc0887d7f7518571eadccdec2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb1dc3c0814cbedc0887d7f7518571eadccdec2f"}]},"branch":"refs/heads/master"},"b2ba48aff09216545c4b35153092222e427c2fca":{"kind":"REWORK","_number":8,"created":"2020-09-18 11:35:07.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/8"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-18 11:34:36.000000000","tz":180},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2ba48aff09216545c4b35153092222e427c2fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2ba48aff09216545c4b35153092222e427c2fca"}]},"branch":"refs/heads/master"},"9d8ae4c92d3762238bc95e17e0f365b5eda4676c":{"kind":"REWORK","_number":9,"created":"2020-09-18 13:49:54.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/9"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-18 13:47:37.000000000","tz":180},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d8ae4c92d3762238bc95e17e0f365b5eda4676c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d8ae4c92d3762238bc95e17e0f365b5eda4676c"}]},"branch":"refs/heads/master"},"8b0a12d71f412fb81ade51a8a53fc5f9f0fa547e":{"kind":"REWORK","_number":10,"created":"2020-10-02 10:16:09.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/10"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-10-02 10:14:37.000000000","tz":180},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b0a12d71f412fb81ade51a8a53fc5f9f0fa547e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b0a12d71f412fb81ade51a8a53fc5f9f0fa547e"}]},"branch":"refs/heads/master"},"c2e173ec7109ddcc405aba7bb13d015a31fdd603":{"kind":"REWORK","_number":11,"created":"2020-10-02 15:15:54.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/55/722655/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/55/722655/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/55/722655/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/55/722655/11"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-30 14:36:31.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-10-02 15:15:40.000000000","tz":180},"subject":"WIP. Refactor backup configuration","message":"WIP. Refactor backup configuration\n\nChange-Id: If63b90e80b38c1995ddf364520be2dec34ed9dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2e173ec7109ddcc405aba7bb13d015a31fdd603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2e173ec7109ddcc405aba7bb13d015a31fdd603"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29568,"name":"Alex Deiter","display_name":"Alex Deiter","email":"alex.deiter@gmail.com","username":"deiter"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
