)]}'
{"id":"openstack%2Fcinder~723825","triplet_id":"openstack%2Fcinder~stable%2Ftrain~I693980384df22b2fa581d8715f73c69b0598dd59","project":"openstack/cinder","branch":"stable/train","topic":"bp/vxflexos-replication-support","hashtags":[],"change_id":"I693980384df22b2fa581d8715f73c69b0598dd59","subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","status":"ABANDONED","created":"2020-04-28 07:38:00.000000000","updated":"2020-07-01 16:01:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0da5f0ff9467d6d7e12633a7e0bb4697fd1c5119","_number":723825,"virtual_id_number":723825,"owner":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-05 20:01:53.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"date":"2020-06-05 21:04:22.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"date":"2020-06-06 16:47:56.000000000","_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"date":"2020-06-10 05:18:34.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"date":"2020-06-05 21:35:26.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"date":"2020-06-09 17:07:08.000000000","_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"date":"2020-06-05 20:22:37.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"date":"2020-06-05 20:57:02.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-05 21:54:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"date":"2020-06-06 04:16:37.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"date":"2020-06-11 10:33:35.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"date":"2020-06-06 01:43:41.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"date":"2020-06-06 06:40:19.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"date":"2020-06-05 23:00:33.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"disliked":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":-1,"date":"2020-06-10 19:57:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-06 02:01:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 09:30:00.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2020-05-27 10:32:13.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-27 11:31:10.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-27 11:46:00.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-05-29 12:25:21.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-06-05 20:01:53.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-05 20:22:37.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-05 20:57:02.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-06-05 21:04:22.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-06-05 21:35:26.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-05 21:54:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-05 23:00:33.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-06-06 01:43:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-06 02:01:24.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-06-06 04:16:37.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-06 06:40:19.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-06 16:47:56.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-06-10 05:18:34.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 19:57:34.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-11 10:33:35.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"}],"messages":[{"id":"840e864b63de855f01c8f163a3a9f6e169b4ff5c","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-04-28 07:38:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41ce526e37c73ad4de4aadaf0af10518a875a2be","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-28 08:00:12.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/723825/1/iscsi/ : FAILURE in 21m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/723825/1/smb/ : FAILURE in 20m 54s","accounts_in_message":[],"_revision_number":1},{"id":"90aa7e9adfeedd9a3ee546b2a82e0fd2e8c587a6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-28 08:15:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/25/723825/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ec3ebba/ec3ebba.html : POST_FAILURE in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"66dddecd167e77b366c2a9ca37105cf774f443c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 09:03:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69c904b6929f44598e3681767ef5f8eb : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea421984029d409d83701f4b9fd0625d : SUCCESS in 9m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/584febde29cd45038de835af0cd6d561 : FAILURE in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3272cb4e6c8c414c9e9e398150df661f : SUCCESS in 13m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ca54d300b4d413a976e6fd0394f347a : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f78d4f5e104a5ba5a90906e23fcc71 : SUCCESS in 14m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/219823a9bf8a4684a519536137c289f1 : EXCEPTION\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/098588bb27944b89be868018bfdee9f8 : EXCEPTION\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01373ad1cc3345669c609e5f46ffe800 : EXCEPTION\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/067bca5d43434f8a8f33643f051cff04 : SUCCESS in 23m 51s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2e2af2b7d5844a299a5c821fb940c96b : EXCEPTION\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c713b7497266498db8bd75646be83b1d : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97fcd83655bd47009cf521f5d2561dd6 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/513cc58d4a1d40328211114a2d24d2ba : FAILURE in 21m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7d9bce164a3543c48de39156a46b1c34 : SUCCESS in 7m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e956174cff2f4ad2994b95be8b06835c : EXCEPTION (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/42bb2c3cdf8a437691c13c1282341261 : SUCCESS in 6m 27s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fac0a2d20c0c491d9a567da51f6219e5 : EXCEPTION\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82f2ac22d0d94ce4b34cb19f5c55e9c7 : EXCEPTION (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/e285153cd3cc4e44a44d947283e62f30 : EXCEPTION (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/174141dc67a04d3d8a6c38ae275d503e : EXCEPTION\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c34d62b955134fe7857da6862a677d32 : EXCEPTION\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f7dd1e5688d4e5f86e8b860ce418897 : EXCEPTION","accounts_in_message":[],"_revision_number":1},{"id":"a4d2554cd6df2d693c75bd0b486ab642a5ce5f82","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-28 09:54:05.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/25/723825/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/262ed5b : FAILURE in 1h 52m 34s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/25/723825/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/193f371 : SUCCESS in 1h 20m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"7e9865ac958c5ecdc8d1596cd950bd10860da0b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-28 09:58:21.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/1/check/tempest-dsvm-full-bionic/7ef063f/ : FAILURE in 53m 25s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/1/check/devstack-plugin-ceph-dsvm-bionic/5d12b85/ : SUCCESS in 2h 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ae235e9209adeee98f9ad37e3961cc2616d7e1be","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-28 11:47:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 1h 01m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7d3ee1659803e031533de68129476b4a01d38f46","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-28 12:33:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/867/index.html : SUCCESS in 1h 13m 40s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1016/index.html : SUCCESS in 1h 51m 01s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1055/index.html : FAILURE in 1h 32m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fa40b55c4720cbf9a77d39849b469dce0f3598ca","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 17:56:55.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"9190c555f16110cb1254e3b54c6d21326b2423bb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-04-28 17:57:21.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"b01d43d0d5726fa39b830eac280bee79e59bccef","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-04-30 09:13:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f179c1a9854221c5781794d13c32e5dad47e50b","author":{"_account_id":31708,"name":"Datera CI","email":"g-ecosystems@datera.io","username":"datera"},"date":"2020-04-30 09:19:26.000000000","message":"Patch Set 2:\n\nBuild Unstable \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/153881cc15d31e0178167f56393e5db8f598c1e4/index.html : UNSTABLE","accounts_in_message":[],"_revision_number":2},{"id":"d5e4488d9e93a928c79291a13c87b1b5fa2877ae","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-30 09:40:53.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/723825/2/iscsi/ : FAILURE in 26m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/723825/2/smb/ : FAILURE in 24m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e80ec80225acaab7058a3712681845d6b300acf8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-04-30 10:07:06.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/25/723825/2/check/dsvm-tempest-infinibox-fc/9d4a7ce : FAILURE in 39m 26s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/25/723825/2/check/dsvm-tempest-infinibox-iscsi/f38ccff : FAILURE in 18m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e74619b4464967efadc099cbc68c251edc88e42c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-04-30 10:57:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 47m 29s","accounts_in_message":[],"_revision_number":2},{"id":"fed608e7b46a9d8d7d6d9d4addf4bc7fb89e13ef","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-04-30 11:10:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/25/723825/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2bae67c/2bae67c.html : FAILURE in 1h 50m 11s","accounts_in_message":[],"_revision_number":2},{"id":"348a61bf0741976e92a10492ecc8a54456793639","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-04-30 11:40:03.000000000","message":"Topic changed from bp/vxflexos-revert-to-snapshot to bp/vxflexos-replication-support","accounts_in_message":[],"_revision_number":2},{"id":"ea74297fe5fbd0b3b9021a83e657a771a1c77a73","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-30 11:44:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/25/723825/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/1a59b6d : SUCCESS in 1h 03m 26s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/25/723825/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/fb87468 : SUCCESS in 1h 21m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"a353d1e401cbdfe0ca7852742d9e805fe41e0047","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-30 11:52:37.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/2/check/tempest-dsvm-full-bionic/63da3e7/ : FAILURE in 55m 19s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/2/check/devstack-plugin-ceph-dsvm-bionic/2337b4f/ : SUCCESS in 2h 34m 14s","accounts_in_message":[],"_revision_number":2},{"id":"397641175c1bb18cbeb78e83d013eb16962a24ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 12:05:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69216ee55387451e9a190c0f4f0bb48c : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58cf7e31d77748b1b9e53b3576f5c954 : SUCCESS in 8m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e24b52bcb39d4ea8a748f73a297f265c : SUCCESS in 14m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f2f607e7fc844f396748fbcd1a8a461 : SUCCESS in 11m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc2cdfcb53a04b9b9b2ce01e08058377 : SUCCESS in 15m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d14871bd12a45a2b0a1caa4e3ab4724 : SUCCESS in 15m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fc24b7a4bde64b5288ff8b199ae29da7 : SUCCESS in 1h 28m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c6fa101b998d44268166452017ebae8d : FAILURE in 1h 07m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a147fdcc94e4982b668beb1b4067132 : FAILURE in 1h 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710e8fd46f724ea9a2836195a2c2d037 : SUCCESS in 26m 38s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b3e456b2d664f97b4b84f3b15b5b301 : SUCCESS in 2h 08m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fceac6f63d204f0fbb7096fc9c5a514c : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32ed95606cce4561906194e9f692108c : SUCCESS in 15m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/86faf8ac5ad94bca9e41030e7ce4a2b7 : FAILURE in 36m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae5b772462fe4be2b6d3f4bb2655aae6 : SUCCESS in 11m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4c32e6641fbe44fbabdb46ac0b74a42d : FAILURE in 1h 40m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/62c2cb12441240a38d7fa0065574a8dd : SUCCESS in 12m 04s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/93ffcb454ccc47dbbd684a248c8f1dbe : SUCCESS in 1h 46m 03s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/22e5ed4976ef4a27a3a14f731929b62d : SUCCESS in 1h 16m 09s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/6fe24394622f4ae5a21148e3dfa161f0 : SUCCESS in 2h 03m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48b4db036f7748abb4462227d27b2e28 : FAILURE in 2h 27m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e113939bcda4f2f9842d23f7f92d4df : SUCCESS in 1h 26m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/822048982aac4b85bc9815db6b186b3d : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":2},{"id":"3a7bf08fe377112d925f8b2456a46d1dfafe130c","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-04-30 13:49:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"906ee4047055a23046b2b76dbba4fd63d1589d70","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-04-30 14:53:31.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/723825/2/iscsi/ : SUCCESS in 1h 03m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/723825/2/smb/ : FAILURE in 27m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c6a240180a44a2dbff0df556f6f08826b32e7560","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-30 16:13:02.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/2/check/tempest-dsvm-full-bionic/d77b0bc/ : FAILURE in 52m 18s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/2/check/devstack-plugin-ceph-dsvm-bionic/4079e1e/ : SUCCESS in 2h 10m 32s","accounts_in_message":[],"_revision_number":2},{"id":"bbb794bc768de3ac7d1172bfe1665860009abe4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 16:22:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/075fc74e15d7449083243655da57e1a2 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2374bff4cf674de1bdc52d59ec78f675 : SUCCESS in 10m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/931bbb8559a547439e2b8a66901b72ab : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25a388a298f34ff6a6763186f1a4ab70 : SUCCESS in 10m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3d09549e2c64f81b7e7c6afd3f936e9 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fe6f0367e940d6994cfc6b6c94a841 : SUCCESS in 14m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/78faa70d1c084f69a33fd41ce1dbd0fe : SUCCESS in 1h 40m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c376dd28f7374f6b96ff6b8a39daa015 : SUCCESS in 58m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30e8615581ee41f0a0b020b298e34afe : SUCCESS in 1h 22m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b31fddeada3494c885e267c8cb493fb : SUCCESS in 22m 07s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aa8ab02416514727b14d7cd33d6334ac : SUCCESS in 1h 44m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3b37c5ff806e43a681b71682699d86bb : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec04ecb2d3374965b7c9881871f5f261 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/53dc4f50b61e4b0e89893e759593edd4 : FAILURE in 19m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b3896470fa994d56b3e1e36db9710d25 : SUCCESS in 8m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/374064f4ddca4c45803ff694bbf3b865 : FAILURE in 1h 19m 46s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8f879814591f4c0d8be9ef7628231d94 : SUCCESS in 6m 17s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cfd04a9b54054058a96b543d36f854be : SUCCESS in 1h 05m 25s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a2814f1b5ef84059840ebcd5bd469e7c : SUCCESS in 1h 16m 26s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/2a195ffe6154490e97567d94ebaab525 : SUCCESS in 1h 28m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee2bbd7ae26f49bb972b0265a8b0a4d3 : SUCCESS in 2h 27m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d3a36142ea4a4581abf35742e2fa3cd8 : SUCCESS in 1h 02m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d460bf51bf040469ccd6ad59d51b568 : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":2},{"id":"53d7cb71daba85fd71f236b08223338e3b09b796","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-04-30 16:27:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c03227410f001fa3810da90230e5629a0e219556","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-04-30 17:12:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/886/index.html : SUCCESS in 1h 18m 34s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1035/index.html : SUCCESS in 1h 53m 16s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1074/index.html : SUCCESS in 1h 36m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b3747f3d12da7f3e6367b46a41d867db0a57d234","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-04-30 18:11:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/25/723825/2/upstream-solidfire-check/cinder-solidfire-iSCSI/4c48ce2/ : SUCCESS in 59m 24s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"ba8b498e308f1f9e4e724dc46a680a4c0edc46bf","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-04-30 18:38:26.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"096f7596775fdd7bfd8c79abe2f8224070d6daa3","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-30 20:27:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/25/723825/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/57a0977 : FAILURE in 1h 53m 42s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/25/723825/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/af40e46 : SUCCESS in 1h 08m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"4ed33f49891bc45b6503dae5d83afecc4deaf81f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-30 20:59:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/25/723825/2/upstream-check/cinder-cDOT-iSCSI/2f1e063/ : SUCCESS in 50m 46s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/25/723825/2/upstream-check/cinder-cDOT-iSCSI-IPv6/f6288d4/ : SUCCESS in 2h 35m 31s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/25/723825/2/upstream-check/cinder-cDOT-NFS/a2e147e/ : SUCCESS in 3h 34m 49s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/25/723825/2/upstream-check/cinder-cDOT-NFS-IPv6/4e15f2e/ : FAILURE in 3h 55m 47s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/25/723825/2/upstream-check/cinder-cDOT-FCP/c61527e/ : FAILURE in 2h 56m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d7b726754e8636dc179ff1358f9076058ac25de4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-01 04:03:41.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"616074397676df3eed23a04be0f752cd850159c4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-05-01 09:45:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 9h 57m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d03f322482fd40117bee753ce9aeabf9ae9e19be","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-01 10:12:32.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4c903367abdbb093858d4b8013c3a23c2c988abb","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-05-03 00:23:57.000000000","message":"Patch Set 2: Code-Review+1\n\nClean backport. Need 3rd party CI results.","accounts_in_message":[],"_revision_number":2},{"id":"ef2a398e980746d8cf1137a5e3517467ba9d5936","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-05-06 07:33:38.000000000","message":"Patch Set 2:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":2},{"id":"0cbada6c7099b350731a91214f64ce7a9e34ac23","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-05-09 10:36:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/723825/2/20200509165129/NEC-ISCSI : SUCCESS in 2h 37m 19s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/723825/2/20200509165129/NEC-FC : SUCCESS in 2h 44m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f7fe9f6f73d961d1b364d21419e6d29a06f13551","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-05-20 09:22:33.000000000","message":"Patch Set 2:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":2},{"id":"6c07598a38d5746172616039865510e095577621","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-05-27 04:06:04.000000000","message":"Patch Set 2:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":2},{"id":"63df8be4969aee861e34872f65aa8493e317a001","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-05-27 06:30:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/25/723825/2/check/DellEMC_VxFlexOS/e401495/DellEMC_VxFlexOS/48 : FAILURE in 2h 23m 55s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"bbf1a0e42b430c74043b44701172463a41252fad","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-05-27 09:12:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6bcd38252a247fb0ccc5182c4ba9d44411f27d35","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-05-27 09:19:15.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6618dee29d5bada605a77c6d3c968e55b9952554","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-05-27 09:30:00.000000000","message":"Patch Set 4:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":4},{"id":"007c0571a1c24b237904fe9220370715c7900bca","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-05-27 09:58:09.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/ee88742ec3e8ae4fe3df51db5dde8c5d0e07693b/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"6a1828448ba52d829840edbfef1fe65856146b6d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-05-27 09:59:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/25/723825/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c773a98/c773a98.html : FAILURE in 39m 31s","accounts_in_message":[],"_revision_number":4},{"id":"a778a915f18122a8037ea81aa7ef3ebe429671a3","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-05-27 10:04:42.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/1d657fdd50b67040f379f97cb586126f2114aa48/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"7e6b36c82e081792289510dc63c0d90e8acde10c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-05-27 10:29:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/25/723825/4/check/dsvm-tempest-cisco-zonemanager-job/252bd27/ : SUCCESS in 1h 04m 51s","accounts_in_message":[],"_revision_number":4},{"id":"48cbd630d59c09c3551e21d60c8d11c67e93130e","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-05-27 10:32:13.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/25/723825/4/check/Cinder-AS13000/4b05343 : FAILURE in 10m 46s\n- cinder-test-fc http://218.29.111.78:9000/25/723825/4/check/cinder-test-fc/ce88e00 : FAILURE in 1h 02m 25s\n- cinder-test-iscsi http://218.29.111.78:9000/25/723825/4/check/cinder-test-iscsi/f701a3a : FAILURE in 7m 20s","accounts_in_message":[],"_revision_number":4},{"id":"a482a07ae6e0aa6b591200d630a1083fb3830722","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-05-27 10:33:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/723825/4/iscsi/ : SUCCESS in 1h 05m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/723825/4/smb/ : SUCCESS in 1h 13m 11s","accounts_in_message":[],"_revision_number":4},{"id":"fc64c8643d4595556520379195051ef44ce0cd72","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-27 11:25:27.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/4/check/tempest-dsvm-full-bionic/5e0e288/ : FAILURE in 26m 05s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/4/check/devstack-plugin-ceph-dsvm-bionic/2b49023/ : SUCCESS in 1h 40m 20s","accounts_in_message":[],"_revision_number":4},{"id":"d50e378acffbf965565c7bb088a444d0c0d7c98a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:27:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/419149f6ad75411fbd4740db0b8e9350 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b4b40a66244a3daf2a830fc24e176d : SUCCESS in 11m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da43168cc16b4b40ac2fb0033abca72d : SUCCESS in 14m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/463ab8f852fe4ade989788ff2d1275fc : SUCCESS in 10m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa079feb45af41a0aada41811b1565c2 : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99943e829c5b46abbf6e273aa4c01229 : SUCCESS in 20m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1efbc6e0a7e8438e9e90b05683b9745b : SUCCESS in 1h 33m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1eeee6757d16496f8bc326e68e6771ad : SUCCESS in 1h 06m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2d9349ab1c274bd99e9cfbbdf0cbcfc7 : SUCCESS in 1h 08m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a8fb5fab11744b9ae6e9e0df9568cf5 : SUCCESS in 11m 48s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0fe70f98d722405b8e6bcd5adebe9c80 : SUCCESS in 1h 31m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e03b218cd914498489b98324398896a5 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45d678317d4f46a6ba8eb9eb10e97fb9 : SUCCESS in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8955246037524edcaebdea21163bfd6e : FAILURE in 19m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f923249a4f82431da36a108f55958a48 : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1093db8463df4ad1988b44dc4934af61 : FAILURE in 1h 22m 51s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e925f5df8a1f42488ce9e9cb0ed13ade : SUCCESS in 7m 58s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f93a84b012b94f7e81e4ed64794eca34 : SUCCESS in 1h 38m 55s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e350af0f28934c948ae69d081db1bce5 : SUCCESS in 1h 22m 25s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/c9797e0f4d5c4236b70a39e5ff4b3fe0 : SUCCESS in 1h 32m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9446024534b548ef8c8d7064782c1755 : SUCCESS in 2h 06m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/492cb023cbd44b1a81fb1e9afe52b664 : SUCCESS in 1h 04m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1069733e346b4411bc3fdbe7cf0113fd : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":4},{"id":"a121fee4d530f1258392d239c8977c1d2c7f2ba1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-27 11:28:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/25/723825/4/upstream-check/cinder-cDOT-iSCSI/5b8b049/ : NOT_REGISTERED\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/25/723825/4/upstream-check/cinder-cDOT-iSCSI-IPv6/dad11d4/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/25/723825/4/upstream-check/cinder-cDOT-NFS/cee8e2d/ : NOT_REGISTERED\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/25/723825/4/upstream-check/cinder-cDOT-NFS-IPv6/2e64aad/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/25/723825/4/upstream-check/cinder-cDOT-FCP/28d6507/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"a9082f847bca5a8c95138ce00097e81fa6e06559","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-05-27 11:28:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/25/723825/4/upstream-solidfire-check/cinder-solidfire-iSCSI/81baf79/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"c9e55c8919f917ff10434377beb89a2f41502af6","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-05-27 11:31:10.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/25/723825/4/check/synology-dsm-cinder-iscsi/cdb9e7a : FAILURE in 2m 53s","accounts_in_message":[],"_revision_number":4},{"id":"5ba50fe2cb8b330ff71ee2a0f4248f0547b02d3c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-27 11:46:00.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/25/723825/4/check/dsvm-tempest-infinibox-fc/d83150c : FAILURE in 28m 31s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/25/723825/4/check/dsvm-tempest-infinibox-iscsi/942f289 : FAILURE in 19m 07s","accounts_in_message":[],"_revision_number":4},{"id":"56791a319f4a28adec249fb3e0364c5bf200d3f0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-05-27 11:54:23.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/25/723825/4/check/DellEMC_VxFlexOS/36139e8/DellEMC_VxFlexOS/49 : SUCCESS in 2h 24m 02s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"dcede85aaa7c8ea285b808cd163caa57095a47f8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-27 14:22:44.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/25/723825/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4466a7f : FAILURE in 32s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/25/723825/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/69a275b : FAILURE in 11s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"81ed7faee33b2b353dfdf592aaa319f26acff2f8","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-05-27 15:32:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1083/index.html : SUCCESS in 1h 18m 29s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1234/index.html : FAILURE in 19s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1261/index.html : SUCCESS in 1h 33m 49s","accounts_in_message":[],"_revision_number":4},{"id":"0b9e3140280d54f94706d52861e0e93a34d46c3d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-27 21:10:25.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"29769c6507dee97a3fa547c041e1fffb17e585ed","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-05-27 21:21:23.000000000","message":"Patch Set 4: Code-Review+1\n\nclean backport","accounts_in_message":[],"_revision_number":4},{"id":"f927de2478139b9bdbef25dcb9537d6bd991d826","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-05-27 22:06:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/723825/4/20200528043527/NEC-ISCSI : SUCCESS in 2h 10m 08s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/723825/4/20200528043527/NEC-FC : SUCCESS in 2h 30m 55s","accounts_in_message":[],"_revision_number":4},{"id":"af1f30742f017adc9081e8098bfb176052c6caf7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-27 23:02:16.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9d458221ab3bd9d761d24c9e9bf20f89fd19ef4e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-27 23:39:46.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"87a72ab3c03154552dcc47012b647740675cf693","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-28 10:38:54.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3f4c5d4e6404c8761290135bdfee195edc1d188a","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-05-29 12:25:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9bdbfaeb1272f473a688dd2fe56a38a0044e3ab3","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-05 19:40:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"71e2e74b525ef76c05607e2a85832069bd2b3ed6","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-05 19:47:11.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/63636496f271a4880291c886e8ea27488ae6b056/index.html : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"43e535612a649f96e55313add6eeb972742997c6","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-05 19:54:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ad5f8840e61b0c645df354b4e0e20523808a4c0b","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-05 20:01:53.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/d6ea4ad8de1052b2f55100e8e7c6860f980e7e11/index.html : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"b3e56f11005c5e79445a9c70f9220430e71c9489","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-05 20:22:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/25/723825/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/389b796/389b796.html : FAILURE in 26m 28s","accounts_in_message":[],"_revision_number":6},{"id":"4138e28862e6e5b6000b482fbf1510b97d534281","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-05 20:47:17.000000000","message":"Patch Set 6:\n\nNew patchset contains fix for bug #1823200 which is already merged in\nstable/train.","accounts_in_message":[],"_revision_number":6},{"id":"3580038ffd94ec9aa3ac47d1d09d105619e9fb0c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-05 20:57:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/723825/6/iscsi/ : SUCCESS in 58m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/723825/6/smb/ : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":6},{"id":"fca223a664bbef8b32e7fa4fe54ec1eb8ce2de3e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-05 21:04:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/25/723825/6/check/dsvm-tempest-cisco-zonemanager-job/3601878/ : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":6},{"id":"7bfd7cb982f69cb3e0111b4d010a6eb8d7622f14","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-05 21:35:26.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/6/check/tempest-dsvm-full-bionic/8ca4c86/ : FAILURE in 30m 21s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/25/723825/6/check/devstack-plugin-ceph-dsvm-bionic/6441d62/ : SUCCESS in 1h 35m 06s","accounts_in_message":[],"_revision_number":6},{"id":"a2db76b8c3608d4c0fe6a3902507a7ef019e4cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:54:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2a091d6ae8147e9836bfd8cf1e5f5ad : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86434dc71c5943fab051f524a36d5aec : SUCCESS in 10m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/455418f05ae44efdbfc6c56f082e4c1b : SUCCESS in 10m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23712d1427ea4c28b6c1d8f5b6862a04 : SUCCESS in 8m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b81e86c760b4e92815b42c61695a4c3 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696c943d3bd741b2a36982781eabf10d : SUCCESS in 14m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/74622bc8e1b746619e9946b8592563ef : SUCCESS in 1h 45m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4ef25962ba9143929aa9dcdfe87e599b : SUCCESS in 48m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a2df1c390524ac0b9aa21d4bbf1c3ab : SUCCESS in 51m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d7efb8f4ffe4ae696db6e4eff849ba5 : SUCCESS in 10m 19s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af8436bcbd9d493695a04b6f83448504 : SUCCESS in 1h 02m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/28338bf651084e7c9759eccef8222b33 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50b9d9ca9319422aa360ffa4f656fa90 : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/540ebc4e31e84fe0a728c8a8e068e045 : FAILURE in 13m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37fdac5298584db191ca4c204f8b0573 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c0c5e04368c47c9b7fd879bd4cbfa65 : FAILURE in 50m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f2f7c60404a849c7a194cbed5b564692 : SUCCESS in 5m 16s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c39b31646dd407b9fe10577ad8bef7e : SUCCESS in 46m 16s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ead2c19762a4190bbf3ef31a091be52 : SUCCESS in 59m 27s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/f24f252914ed4f2e9cd8727e58823a5f : SUCCESS in 1h 01m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e601feb96304063b5ff955cd710ef55 : SUCCESS in 1h 58m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2c819e703fce49518ba376be26079959 : SUCCESS in 55m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a760022c7507443181d5e84b2a16acfa : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":6},{"id":"2afd97c05956a8fcbb48975121944a099a07b865","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-05 23:00:33.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/25/723825/6/upstream-solidfire-check/cinder-solidfire-iSCSI/a60cb38/ : SUCCESS in 1h 05m 09s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"639106b1e5f54e3836428cd92d55ecfa286f8a0c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-06 01:43:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/25/723825/6/upstream-check/cinder-cDOT-iSCSI/9e65855/ : FAILURE in 3h 48m 32s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/25/723825/6/upstream-check/cinder-cDOT-iSCSI-IPv6/f8544ac/ : SUCCESS in 2h 20m 58s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/25/723825/6/upstream-check/cinder-cDOT-NFS/67e15dd/ : FAILURE in 1h 48m 26s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/25/723825/6/upstream-check/cinder-cDOT-NFS-IPv6/ad3870e/ : SUCCESS in 2h 16m 50s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/25/723825/6/upstream-check/cinder-cDOT-FCP/2c48ee3/ : SUCCESS in 1h 24m 59s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"7730a9380042e994fcdf2e7aed3da1a53851adf8","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-06-06 02:01:24.000000000","message":"Patch Set 6: Code-Review+1\n\nneed CI results.","accounts_in_message":[],"_revision_number":6},{"id":"dc5f4d510affcf3f9296a88f3b7abdf3b8545d88","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-06 04:16:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/25/723825/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/314c6f1 : SUCCESS in 1h 00m 40s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/25/723825/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/992dd6b : SUCCESS in 1h 11m 42s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"23b6a9f545e6d2974c375cf15d2edd8de3b5f758","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-06-06 06:40:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/723825/6/20200606130755/NEC-FC : SUCCESS in 2h 30m 04s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/723825/6/20200606130755/NEC-ISCSI : SUCCESS in 2h 32m 22s","accounts_in_message":[],"_revision_number":6},{"id":"5e0b843a4f9bb3f219bfc145c6c73167210f1484","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-06 16:47:56.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1159/index.html : SUCCESS in 1h 13m 59s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1310/index.html : FAILURE in 19s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1337/index.html : SUCCESS in 1h 30m 23s","accounts_in_message":[],"_revision_number":6},{"id":"a9f4aff8fff5d55408089aeba61dcf58c830d41e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-07 10:02:59.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"290a5b81e761c2f2aca6fe7dbb8bc990cefe8247","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-07 11:25:13.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"0aa59af68c97b6efe16a5891f881be64b77669ba","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-07 16:59:28.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"1c0a36f8797e41a0cddcde7d4790d1440cde7821","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-09 17:07:08.000000000","message":"Patch Set 6:\n\nrun-DellEMC VxFlexOS CI","accounts_in_message":[],"_revision_number":6},{"id":"2308b4a93eb16bff24dac1709dcbfac1d8e23b7b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-10 05:18:34.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/25/723825/6/check/DellEMC_VxFlexOS/592ee3c/DellEMC_VxFlexOS/117 : ABORTED in 3h 19m 53s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"ef32d15c4620d3c8dbc35923dc2386d3e7e72a55","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-10 19:57:34.000000000","message":"Patch Set 6: Code-Review-1\n\nI\u0027m not sure that it\u0027s a good idea to backport it. I\u0027m leaving -1 here but other stable-cores still can +2 on it","accounts_in_message":[],"_revision_number":6},{"id":"691a006ad5686fc237ea0a13dd37a3006fd256c5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-11 10:33:35.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/25/723825/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"0da5f0ff9467d6d7e12633a7e0bb4697fd1c5119","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-01 16:01:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d6ea4ad8de1052b2f55100e8e7c6860f980e7e11","revisions":{"eefda5ff970fde3ae451542f1bd0a02d7054cd3c":{"kind":"REWORK","_number":1,"created":"2020-04-28 07:38:00.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/25/723825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/723825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/723825/1"}}},"commit":{"parents":[{"commit":"b91f5146e04827cdd9340b9b6231416563d56281","subject":"Fix VxFlexOS documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b91f5146e04827cdd9340b9b6231416563d56281"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-01-31 07:51:39.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-04-28 07:37:42.000000000","tz":180},"subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","message":"Add support for VxFlex OS 3.5 to VxFlex OS driver\n\nDriver code is prepared for future VxFlex OS 3.5 release.\n\nUnit tests are fixed to work properly with updated driver.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I693980384df22b2fa581d8715f73c69b0598dd59\n(cherry picked from commit f75b2865fe2aee3f6aefea8f08ce92e1d590ac98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eefda5ff970fde3ae451542f1bd0a02d7054cd3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eefda5ff970fde3ae451542f1bd0a02d7054cd3c"}]},"branch":"refs/heads/stable/train"},"153881cc15d31e0178167f56393e5db8f598c1e4":{"kind":"REWORK","_number":2,"created":"2020-04-30 09:13:54.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/25/723825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/723825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/723825/2"}}},"commit":{"parents":[{"commit":"0aceffbf9eed12c91fe8b8760757d8f11a3d08d3","subject":"Merge \"Fix: failed to create snapshot with DriverFilter\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0aceffbf9eed12c91fe8b8760757d8f11a3d08d3"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-01-31 07:51:39.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-04-30 09:12:30.000000000","tz":180},"subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","message":"Add support for VxFlex OS 3.5 to VxFlex OS driver\n\nDriver code is prepared for future VxFlex OS 3.5 release.\n\nUnit tests are fixed to work properly with updated driver.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I693980384df22b2fa581d8715f73c69b0598dd59\n(cherry picked from commit f75b2865fe2aee3f6aefea8f08ce92e1d590ac98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/153881cc15d31e0178167f56393e5db8f598c1e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/153881cc15d31e0178167f56393e5db8f598c1e4"}]},"branch":"refs/heads/stable/train"},"ee88742ec3e8ae4fe3df51db5dde8c5d0e07693b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-27 09:12:55.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/25/723825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/723825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/723825/3"}}},"commit":{"parents":[{"commit":"ac6e0c472fb6276afbd6d421c410ccc844369563","subject":"[stable only] Add warning about rbd_keyring_conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac6e0c472fb6276afbd6d421c410ccc844369563"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-01-31 07:51:39.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-05-27 09:12:55.000000000","tz":0},"subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","message":"Add support for VxFlex OS 3.5 to VxFlex OS driver\n\nDriver code is prepared for future VxFlex OS 3.5 release.\n\nUnit tests are fixed to work properly with updated driver.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I693980384df22b2fa581d8715f73c69b0598dd59\n(cherry picked from commit f75b2865fe2aee3f6aefea8f08ce92e1d590ac98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee88742ec3e8ae4fe3df51db5dde8c5d0e07693b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee88742ec3e8ae4fe3df51db5dde8c5d0e07693b"}]},"branch":"refs/heads/stable/train"},"1d657fdd50b67040f379f97cb586126f2114aa48":{"kind":"REWORK","_number":4,"created":"2020-05-27 09:19:15.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/25/723825/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/723825/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/723825/4"}}},"commit":{"parents":[{"commit":"ac6e0c472fb6276afbd6d421c410ccc844369563","subject":"[stable only] Add warning about rbd_keyring_conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac6e0c472fb6276afbd6d421c410ccc844369563"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-01-31 07:51:39.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-05-27 09:19:01.000000000","tz":0},"subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","message":"Add support for VxFlex OS 3.5 to VxFlex OS driver\n\nDriver code is prepared for future VxFlex OS 3.5 release.\n\nUnit tests are fixed to work properly with updated driver.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I693980384df22b2fa581d8715f73c69b0598dd59\n(cherry picked from commit f75b2865fe2aee3f6aefea8f08ce92e1d590ac98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d657fdd50b67040f379f97cb586126f2114aa48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d657fdd50b67040f379f97cb586126f2114aa48"}]},"branch":"refs/heads/stable/train"},"63636496f271a4880291c886e8ea27488ae6b056":{"kind":"REWORK","_number":5,"created":"2020-06-05 19:40:24.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/25/723825/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/723825/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/723825/5"}}},"commit":{"parents":[{"commit":"aadf0950c9c62f68a107c89c5b7c7632a2c732db","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aadf0950c9c62f68a107c89c5b7c7632a2c732db"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-01-31 07:51:39.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-05 19:38:16.000000000","tz":180},"subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","message":"Add support for VxFlex OS 3.5 to VxFlex OS driver\n\nDriver code is prepared for future VxFlex OS 3.5 release.\n\nUnit tests are fixed to work properly with updated driver.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I693980384df22b2fa581d8715f73c69b0598dd59\n(cherry picked from commit f75b2865fe2aee3f6aefea8f08ce92e1d590ac98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63636496f271a4880291c886e8ea27488ae6b056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63636496f271a4880291c886e8ea27488ae6b056"}]},"branch":"refs/heads/stable/train"},"d6ea4ad8de1052b2f55100e8e7c6860f980e7e11":{"kind":"REWORK","_number":6,"created":"2020-06-05 19:54:57.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/25/723825/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/723825/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/723825/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/723825/6"}}},"commit":{"parents":[{"commit":"aadf0950c9c62f68a107c89c5b7c7632a2c732db","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aadf0950c9c62f68a107c89c5b7c7632a2c732db"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-01-31 07:51:39.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-05 19:54:44.000000000","tz":180},"subject":"Add support for VxFlex OS 3.5 to VxFlex OS driver","message":"Add support for VxFlex OS 3.5 to VxFlex OS driver\n\nDriver code is prepared for future VxFlex OS 3.5 release.\n\nUnit tests are fixed to work properly with updated driver.\n\nImplements: blueprint vxflexos-replication-support\nChange-Id: I693980384df22b2fa581d8715f73c69b0598dd59\n(cherry picked from commit f75b2865fe2aee3f6aefea8f08ce92e1d590ac98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6ea4ad8de1052b2f55100e8e7c6860f980e7e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6ea4ad8de1052b2f55100e8e7c6860f980e7e11"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
