)]}'
{"id":"openstack%2Fcinder~725135","triplet_id":"openstack%2Fcinder~stable%2Ftrain~I0b8b5930c9e8f3c049b82e92c5fb7b05d6ae9291","project":"openstack/cinder","branch":"stable/train","topic":"bug/1839384","hashtags":[],"change_id":"I0b8b5930c9e8f3c049b82e92c5fb7b05d6ae9291","subject":"NetApp ONTAP: Fix iSCSI multiattach volume terminates connection","status":"MERGED","created":"2020-05-03 18:10:36.000000000","updated":"2020-06-27 02:21:19.000000000","submitted":"2020-06-27 01:46:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725135-1593222386453-9fea2b71","meta_rev_id":"707d30e7abd1e41afc0494a8decb3843cfe25153","_number":725135,"virtual_id_number":725135,"owner":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-03 18:22:00.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-26 15:36:58.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2020-05-04 23:51:46.000000000","_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2020-06-24 19:43:28.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-03 19:33:33.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-06-24 15:32:23.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"date":"2020-06-26 22:59:52.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"date":"2020-06-26 21:59:33.000000000","_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"date":"2020-05-05 14:34:41.000000000","_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-27 01:46:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-27 00:06:09.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-26 20:55:16.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-27 00:59:08.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-05-04 15:35:47.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"date":"2020-05-10 19:41:23.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":2,"date":"2020-06-19 15:25:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-04 15:39:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 18:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":1,"date":"2020-06-19 15:25:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-03 18:22:00.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-03 19:33:33.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-05-04 15:35:47.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-04 15:39:24.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-04 23:51:46.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-05-05 14:34:41.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-05-10 19:41:23.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-15 18:57:01.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-06-19 15:25:46.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-24 15:32:23.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2020-06-24 19:43:28.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-26 15:36:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-06-26 20:55:16.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-26 22:59:52.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-06-27 00:06:09.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-27 00:59:08.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-06-27 01:46:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acd570ed1b7ede665d8413f572c03ed7c22244e6","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-05-03 18:10:36.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"8e1598c64ffe17d7c0c1bdb3bcc110236f0025c9","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-05-03 18:11:32.000000000","message":"Topic changed from bug/1839384-stable/ussuri-stable/train to bug/1839384","accounts_in_message":[],"_revision_number":1},{"id":"390274dbe6b9a8391f33a44bb9d725df828805bf","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-05-03 18:22:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/b9c6d4a7f7ac4bab4f30dc6a323e7eed9863a930/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9964b1993d58d6c2fa3c04376dfac09458906916","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-05-03 19:06:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project review.opendev.org/openstack/nova in 42m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4c1a17600f5fe39d5d08d773874f4f099d9d76e2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-03 19:20:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/35/725135/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/d54f675 : SUCCESS in 59m 59s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/35/725135/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/1fa2945 : SUCCESS in 1h 09m 09s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"b37e3c1ce44fd6701a85b1c7daf81471d056b1d3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-03 19:33:33.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/35/725135/1/check/dsvm-tempest-infinibox-fc/1148768 : FAILURE in 22m 24s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/35/725135/1/check/dsvm-tempest-infinibox-iscsi/6d793b6 : FAILURE in 21m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c4748aa1dcb1715c0ad0a4dfc444aea715ca1b28","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-05-03 19:36:48.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/725135/1/iscsi/ : FAILURE in 1h 25m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/725135/1/smb/ : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":1},{"id":"858c42d3335ea39a306680265004f26c18657d0d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-03 20:12:27.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/725135/1/check/tempest-dsvm-full-bionic/1b21ca9/ : FAILURE in 41m 35s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/725135/1/check/devstack-plugin-ceph-dsvm-bionic/26a1499/ : SUCCESS in 1h 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d38c26cc337b953cc7039f07445dd67846bcd003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 20:40:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a256729853a4803b8addcf04498be17 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97dfd6e673af47b4958c76310ea742df : SUCCESS in 8m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcb37c8e109c427e9c2c5f5dea015628 : SUCCESS in 10m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c96e8adb03f54c4982160adbe627b95a : SUCCESS in 10m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc64b1f424e848a5a7450013539d54ba : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e15404b371c4387a32a53ea5f166bb7 : SUCCESS in 11m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6f3c8e924f214f7a9825f19b62a48480 : SUCCESS in 1h 31m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f77d50e4918846d080bf3da509f34d0c : SUCCESS in 58m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09cf9940a5094aa2aa80c59d11e20941 : SUCCESS in 1h 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/626f27148fa4411ca348432cac720e26 : SUCCESS in 23m 34s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e791d27cba194ddb8d2f7ac35bdda0b5 : SUCCESS in 1h 32m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3dd6b9aceb94d33aaf4dc44fbb282ed : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7b47e2d6797408089e1eb7081bb7c72 : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/67d30013863246beb1241a8835279676 : FAILURE in 18m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cbce2cdac27c488bb58aca907b2cf4d4 : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0081597c818f45b194ace8aaf57fc4b0 : FAILURE in 1h 05m 38s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aee746e7838f43e7a18d9d55bef1afc5 : SUCCESS in 4m 49s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1f4b93488ec74950a3c00f8b68a4463f : SUCCESS in 1h 09m 33s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e81c2233d9fb44b085326b61dc68963f : SUCCESS in 1h 24m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/6a98a258a9c14853ac287e77a1f0eca7 : SUCCESS in 1h 21m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9efac5c3591e4b17bcbcc578285f83f4 : SUCCESS in 2h 16m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e63c8f6afaf34a1aace9a8c14f724023 : SUCCESS in 1h 01m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab7e467f0b334b7fb366d0ae7c594b03 : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d41ea11c58842805e40a43b7183f02e8ff43b283","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-03 23:40:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-iSCSI/b91f759/ : SUCCESS in 50m 19s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-iSCSI-IPv6/e232249/ : SUCCESS in 1h 28m 47s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-NFS/9e81a36/ : SUCCESS in 50m 05s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-NFS-IPv6/738cd2f/ : SUCCESS in 1h 29m 54s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-FCP/0838aaa/ : FAILURE in 2h 58m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"09d944bb5176085630ff3db507a46a7c0043190f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-04 03:26:16.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"32ce15d33f971eb446d0a7334e9a20f8efdfd480","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-04 05:20:45.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d588e8d95997fe23e6d3600dddc8f0f97bb5d036","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-05-04 11:26:48.000000000","message":"Patch Set 1:\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":1},{"id":"5d03de6549e3c071d287db22b0ace3fcf4a0a719","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-04 15:35:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-iSCSI/a1eee41/ : SUCCESS in 1h 19m 55s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-iSCSI-IPv6/6e3746e/ : SUCCESS in 1h 27m 41s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-NFS/071d698/ : SUCCESS in 46m 40s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-NFS-IPv6/5b37ce1/ : SUCCESS in 1h 32m 27s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/35/725135/1/upstream-check/cinder-cDOT-FCP/141e6c9/ : FAILURE in 1h 21m 44s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"30a9374eb77fc090fcefadbadb34486017061478","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-04 15:39:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d539c05c979467fab14497b09f15b09d4bc5d558","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-05-04 23:51:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/920/index.html : SUCCESS in 1h 20m 27s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1069/index.html : FAILURE in 19s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1108/index.html : SUCCESS in 1h 31m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e560c24cbf4c134bbdb44ef3a3fffd80ce13666c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-05-05 14:34:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd6ec08c59a404e5aa61cbacc67613172cee10f5","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-05-05 21:09:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4455e84290a8388bc4280e3f5550a0ca58f7219c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-05-10 19:41:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/725135/1/20200511021553/NEC-ISCSI : SUCCESS in 2h 13m 29s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/725135/1/20200511021553/NEC-FC : SUCCESS in 2h 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6c768481d84882cb567b9503fbbdfa4afdca343b","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-15 18:57:01.000000000","message":"Patch Set 1: Code-Review+2\n\nI am ok with this backport.  Just looks like there is some additional fixing needed in the future.","accounts_in_message":[],"_revision_number":1},{"id":"69da5074ae1bc9f4676a4ddbfae9098cd486c985","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-06-15 20:44:12.000000000","message":"Patch Set 1:\n\nOpened a bug for the issue eharney observed [0]. We might work on that in a near future.\n\n[0] https://bugs.launchpad.net/cinder/+bug/1883603","accounts_in_message":[],"_revision_number":1},{"id":"670c70616928ba869b9c53d0bd491a87d0ea8f1f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-19 15:25:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3d5997de2962373ac8fcfa776ab65b93e0d1966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 15:26:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bc2f27169ef7b5c2f7ece8179d6eada6aeac048","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 16:44:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/284cf10636f04fc48ad3932d7f9ef3c6 : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24f022cb8fc4f938c71cc3d52002401 : SUCCESS in 10m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74cb1e2565b1423aa3b7efc3ffec18e4 : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6acb07141dbf4110945d28a900b0acba : SUCCESS in 12m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc698d370f2d4193ad6c847eb5f13e6e : SUCCESS in 17m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/432fdfc54b9d4133938a6eb21289cc34 : SUCCESS in 18m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba93003891be40b19ffd42765ba5a249 : SUCCESS in 1h 13m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/defc6a33d13b4d1c83924926ee916689 : SUCCESS in 40m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/c2aaf537c95740e38133b87601c4979a : SUCCESS in 43m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c8f9eec64d44a4c974797a6b87191ac : SUCCESS in 1h 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a73e8a5933e4c8a983f2b35894f9024 : SUCCESS in 9m 56s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf78b873a3604f1bbbda79122a5f758d : FAILURE in 48m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9ae0d7fb7eca44c8a79eadde3f11dd56 : SUCCESS in 1h 02m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31265fbcc82f4de3b0e837d70e8e6691 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":1},{"id":"bea4c4297bcc2037e2aae8ddf8c45b9aef9f8f4e","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-06-23 20:43:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d5bae46592bab3f484de4996e09c8d154adfcbd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-23 21:39:25.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"dbcd41237553aca4f41d2a1591a47760a8be41db","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-23 21:46:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/725135/1/iscsi/ : SUCCESS in 1h 02m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/725135/1/smb/ : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":1},{"id":"567da589f4195ab181ed855004d98da5fc0390c3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-24 00:48:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/725135/1/check/dsvm-tempest-cisco-zonemanager-job/3360d44/ : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0bc2c24d4014791fda7f77ac2f2f72a7b003ab3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:37:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96b42e98c4fe45c2a9e6c7f8d12f4952 : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c7c01df0cc4a648c1f1ab5a86699da : SUCCESS in 10m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54d351f73f5344e3818b3c8445edc1a0 : SUCCESS in 10m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4d880f19a7f4a6d922c6a88a59a4e6b : SUCCESS in 9m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6a64013afca4a0197207f644e5425f1 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d081cee1cb8d4496b233d88821cb5e07 : SUCCESS in 12m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6434295b19c14201a2f4d680e90b5911 : TIMED_OUT in 2h 03m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2b9e5f6f511d4c54bce3375fef6faa69 : SUCCESS in 51m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/67184c0b5dee412d8be7e2dd52ad17b2 : SUCCESS in 1h 00m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/757b46e2928248539961e21309dd409b : SUCCESS in 1h 19m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/545ebdb513324367a6a2fe3d2787d978 : SUCCESS in 8m 29s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c1a36073549e464b82ced86fb5590bf3 : FAILURE in 1h 29m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60e89514731b4604ab11f76400d5eb00 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87e497d1cb8c4cb2865f9d494d56d963 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f8457ff14a4a48518ee89f35fa7c05e1 : FAILURE in 21m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/074bb11e30a341b4aad1934f61079a99 : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a0904f7cd994c3cb0beeb81e8af3718 : FAILURE in 1h 03m 00s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/528bb537e97548299569160c42804dea : SUCCESS in 6m 31s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e4ae7fd1c1d3453e96b36b57cd176240 : SUCCESS in 1h 03m 12s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df54301886854544a924abe3cd4ece20 : SUCCESS in 1h 15m 22s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/34f7a5f2b8e24bea964a035cf6c2c538 : SUCCESS in 1h 26m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8b9e42bac3d47f796777e684ad5b273 : SUCCESS in 2h 26m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e625d65103f5441a8d98d6498e526ae5 : SUCCESS in 57m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec97714a05484779b0e4cb0a025b7362 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e4f9184a8726542907626cfed887509dddb96549","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-24 03:22:09.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/725135/1/check/tempest-dsvm-full-bionic/876b651/ : TIMED_OUT in 2h 37m 45s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/725135/1/check/devstack-plugin-ceph-dsvm-bionic/3c03a0e/ : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3a85bbeaa39b4e3195de1c7b00b92d6fa399681d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 03:57:55.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2eed4b512b0656262924d1f0e125f61578493a26","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-24 09:14:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/35/725135/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ed684c7 : SUCCESS in 1h 09m 35s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/35/725135/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/8d7dc60 : FAILURE in 2h 05m 40s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"f6fc211f7425bd6010618dfdeefec96690f9c816","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-06-24 15:32:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9a354aa845aff12745333d7b75928f00c5c29d0","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-24 16:45:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/725135/1/iscsi/ : SUCCESS in 1h 00m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/725135/1/smb/ : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3de8ad1e3cd8392c7cac051457eae0e3a53f351f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 17:16:29.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0a85b6176458d963596c84d335f955eade2d5217","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-24 17:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/725135/1/check/dsvm-tempest-cisco-zonemanager-job/7a21f85/ : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e7d73e1ff5ee533a83011fe798b698102b545000","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-24 19:43:28.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/725135/1/check/tempest-dsvm-full-bionic/0449458/ : TIMED_OUT in 2h 39m 39s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/725135/1/check/devstack-plugin-ceph-dsvm-bionic/556167a/ : SUCCESS in 1h 38m 26s","accounts_in_message":[],"_revision_number":1},{"id":"dffe5ca69fac20a48a4b4aa6b2bcc0334afe1859","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-24 20:00:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/35/725135/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e6a2a80 : FAILURE in 1h 53m 13s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/35/725135/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ce22130 : SUCCESS in 1h 07m 56s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"f27621dce8d9fee385cc0ff385464f20e43255d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 21:37:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69720e410b244bff9b4eb66acc4e8a12 : FAILURE in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7c53e0207e4a83ae8ab5bd1e5212d0 : SUCCESS in 15m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c30516039b746eb979651f549c0d860 : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2066dfad3c04e7780553cfb2609f1e5 : SUCCESS in 9m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bc7cdcd407349f78e685b538f5a8ede : SUCCESS in 15m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8304150e511416dab061935bbe7732a : SUCCESS in 16m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4071c818939f42b28bd4b978d4501918 : SUCCESS in 1h 22m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1d3e18f4fb0546a192101066f7c27f75 : SUCCESS in 40m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/5a2c50e0ba984a16b98340bdd723dd6d : SUCCESS in 1h 11m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb0c832e2b95412fa900ec18609eb6bc : SUCCESS in 1h 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/842c4e674b5842d3baac518f94ed8d36 : SUCCESS in 9m 43s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/42e8a34f5eee45748a33b004f691fcd2 : FAILURE in 1h 21m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4cc6c4792aaa47b28d0775f93cb5be9c : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c263b5138214d65b7268118caf4ffaf : SUCCESS in 8m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f912c84f31c44abe8cc6c52bf80cdc01 : FAILURE in 24m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d3af0df689694881b9e8a4b654370a92 : SUCCESS in 9m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8115d0ba9a694a4ea0010fa11eb73b12 : FAILURE in 1h 25m 20s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8a8b1c0acf124243882a2dd1093eebf1 : SUCCESS in 6m 02s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dbddafe75ba6464597a4461e8f0cef4f : SUCCESS in 1h 23m 17s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/55b46b569cab4479bb2b1eb4e89cc058 : SUCCESS in 1h 51m 53s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/e5902817ff294375a2807ca239a3f551 : SUCCESS in 1h 19m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ffa27d24e8f43498952c00b35fb544e : SUCCESS in 1h 54m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7620a82924184ebe8377015dae9717e4 : SUCCESS in 1h 01m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db94052f96ed4c1f8e2ea1e252870201 : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f8155c6280353486fcec70b9fcefd2c77c0aa5ad","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 21:38:23.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"409b4243784e20d6d9277ad64f9353650e026396","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-26 14:27:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ecacb7b4120085620d5f3c3791fedf72e7037ed1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-26 15:33:04.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/725135/1/iscsi/ : FAILURE in 13m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/725135/1/smb/ : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ce5739f3a5b544d3a8f6b89635c2ed55c0bd7031","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-26 15:35:07.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f5caa6d7aaa1482c244a88ff69e512da5a83bfd6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-26 15:36:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/725135/1/check/dsvm-tempest-cisco-zonemanager-job/15b51f4/ : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5c62cb6afee938f9743ba439102d4870384bcab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:53:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1a436f4416d463d85309e8163988a58 : SUCCESS in 25m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11c66d3a4659495aae92587039d4f22f : SUCCESS in 21m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/693bacce10ff491e8f280714020087a5 : SUCCESS in 25m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/929155b4007c4fe49460c7ad21117baa : SUCCESS in 30m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d898ba03c90443588486e62cff1e445 : SUCCESS in 26m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0f7eb204314becada8d1e50ff08f30 : SUCCESS in 31m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a0a8f5b568548429db296868bb5375e : TIMED_OUT in 2h 06m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/360d803554264187ac96c162ec4ff917 : FAILURE in 1h 35m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/47ab79758974405398b1aadb71883a89 : FAILURE in 1h 29m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c5a2913002d43fab464523ecd4aa4d6 : TIMED_OUT in 2h 06m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3531b79e23c492faf5e7d39f5e3f067 : SUCCESS in 18m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/65ad12495cd34afb83bf3b84e033df3f : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6952a03735d54f65b9feb86e65c01392 : SUCCESS in 17m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/38dce62d9964473fbb6987493b9324b3 : FAILURE in 36m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b8f3d9b63704601a1978ceb6f5444a3 : SUCCESS in 18m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bff796431d3f441b9bd75ece291372ce : FAILURE in 2h 02m 43s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0f6945c4b8914de780472369c651e17b : SUCCESS in 17m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4781ced5c8a493d8316a4a35d2d4ebb : SUCCESS in 2h 32m 35s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de1f360a56224008830ea80795bfc81d : SUCCESS in 2h 04m 58s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5f3e04c9897c424c83b10db15a40a429 : SUCCESS in 2h 08m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/206e85d8b71a43318712afe7ebf4d8cb : SUCCESS in 2h 30m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76e22a1699cf4c07a152a1de227065fd : TIMED_OUT in 3h 05m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a362bab4d982418d9fc8e6ef96e5ac1e : FAILURE in 1h 24m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1f25a2aa40241aa842cdc792ce9aa4c : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":1},{"id":"30883d47d8e2b57733cd4e601760ec99577c260c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-26 20:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/35/725135/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b35117c : SUCCESS in 1h 04m 59s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/35/725135/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/dc6ee32 : SUCCESS in 1h 08m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"5dea808b50f09f5133bb12271208d8d7eeef06f1","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-06-26 21:59:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b99a3bd97c836e245be589aeb6a118983372e0ad","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-26 22:59:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/725135/1/iscsi/ : SUCCESS in 59m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/725135/1/smb/ : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":1},{"id":"76380765963fdfd6463555d16b111c6c020436b5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-27 00:03:02.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"905551204f5597fd13c14e920089ce92d92d3751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 00:05:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9bba318b0bb4918b2dc12c3828e5d36 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaab7010786740518c14e8dd9ae540a1 : SUCCESS in 9m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11c31ca9298745e8929e690db6a403d5 : SUCCESS in 10m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c96ca9eab0b94f9d8e4172cae65ec84e : SUCCESS in 9m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df49c780acaa41a495ca80b28361f25b : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e387b8111640d982e0f556a6c09349 : SUCCESS in 13m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c758910d47304d4a973f246d760ba851 : SUCCESS in 1h 30m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/44ac0dcd24ee432cb723f002c10ce350 : SUCCESS in 49m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/d3f04d237e6e4caa8c1b5091e4bc22bb : SUCCESS in 1h 19m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11db85f328a54ebca319d821cd467e59 : SUCCESS in 1h 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a4a1521226e425ca135881d2bcc83e1 : SUCCESS in 7m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88ec961d5fd74de7b3928d8ed33434e2 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23f1272e95eb417e942f16c95eb36447 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c625ebdd27f84173b375a08b54c86850 : FAILURE in 19m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a333a05fdcfd4ab182f688cf83a23a40 : SUCCESS in 4m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7eb2237170ec44e780888d1f630b5cdf : FAILURE in 1h 17m 16s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8dccb11da0354de2a5b5c825cd189a51 : SUCCESS in 5m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a2121bddad174dcfbaa8b3b49e3f9205 : SUCCESS in 1h 31m 54s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7e67b0cd4d0c4bec8475b917de7e40cc : SUCCESS in 1h 07m 50s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d931f83353d84e8cb287e2c07f8b5866 : SUCCESS in 1h 17m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f7aad158b70d4a9ca198a5bab97b5bbc : SUCCESS in 1h 34m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ff9614842d543639594f8f4949f5a91 : SUCCESS in 1h 59m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/33fad3604568402c9d581eb4de5a13bd : SUCCESS in 1h 04m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/858d0f34292f4c60a5c8a9334c7fd6c0 : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d147cd0c12540bf5c3beeb736d6e2fcc2fce7b62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 00:05:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aede08808af112d976df4b61563bd4c69b27274f","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-06-27 00:06:09.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/35/725135/1/check/synology-dsm-cinder-iscsi/e7a4ad3 : FAILURE in 10s","accounts_in_message":[],"_revision_number":1},{"id":"5e9457cbd46b8560d9505500c3fa14d2c9b89d88","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-27 00:59:08.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/35/725135/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4b417dfd9bff41a0adb18e73ab6570a3dbdffc54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 01:46:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"16c005f3ad015ed9518495950b43154fef33383b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 01:46:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37f78461e9214ee79a5930abd4bc64b3 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ba0dfcea1a455eaf545f68fe3c23ac : SUCCESS in 8m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26fe40927de340ce85d1bab7b7ac403c : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da51a3a67aa3478d9ea003898c9e42e9 : SUCCESS in 9m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ae95b6073ec48b6af9521dae722f75f : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f979d3a13f6049e0b7ac718a50483c46 : SUCCESS in 14m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/415093e1a86243a1842d215a5bb3ac93 : SUCCESS in 1h 34m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fd3c5baad8ea4a33b1c35ad61f7bb6b6 : SUCCESS in 1h 04m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7c8141cebcb2424fb66a3c34da150737 : SUCCESS in 56m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bd5f04b37584ed980f79ac058ebb0db : SUCCESS in 1h 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff007abe0941482a817758c52851bc1a : SUCCESS in 6m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8cef9efbcda547d7bf26992b32e4e333 : SUCCESS in 1h 21m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e2442e6726d348a1835e075503db8aba : SUCCESS in 1h 07m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95e5ed21a8bd42909d13fca52fd81c3b : SUCCESS in 48m 34s","accounts_in_message":[],"_revision_number":1},{"id":"14cec17cb60d39f27e08f68e3291f04fc5dd1176","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 01:47:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1cfebebfacd4470a329f15a87145c0e : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bd57f2fee8848eda0b7f0d68127e359 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1},{"id":"707d30e7abd1e41afc0494a8decb3843cfe25153","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-27 02:21:19.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/725135/1/check/tempest-dsvm-full-bionic/a15bc79/ : TIMED_OUT in 2h 40m 36s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/725135/1/check/devstack-plugin-ceph-dsvm-bionic/49971da/ : SUCCESS in 1h 41m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9c6d4a7f7ac4bab4f30dc6a323e7eed9863a930","revisions":{"b9c6d4a7f7ac4bab4f30dc6a323e7eed9863a930":{"kind":"REWORK","_number":1,"created":"2020-05-03 18:10:36.000000000","uploader":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"ref":"refs/changes/35/725135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/725135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/725135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/725135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/725135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/725135/1"}}},"commit":{"parents":[{"commit":"0aceffbf9eed12c91fe8b8760757d8f11a3d08d3","subject":"Merge \"Fix: failed to create snapshot with DriverFilter\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0aceffbf9eed12c91fe8b8760757d8f11a3d08d3"}]}],"author":{"name":"Lucio Seki","email":"lucioseki@gmail.com","date":"2020-03-17 17:25:24.000000000","tz":0},"committer":{"name":"Lucio Seki","email":"lucio.seki@netapp.com","date":"2020-05-03 18:10:36.000000000","tz":0},"subject":"NetApp ONTAP: Fix iSCSI multiattach volume terminates connection","message":"NetApp ONTAP: Fix iSCSI multiattach volume terminates connection\n\nWhen detaching a volume from an instance, NetApp ONTAP driver terminates\nthe iSCSI connection, even if there\u0027re other connections to the LUN from\nthe same initiator. This terminates connections to the LUN from other\ninstances running on the same host.\n\nThis patch fixes this bug by leaving the connection untouched if there\nare other connections from the same initiator. Now the connection is\nterminated only if it\u0027s the last one from the same initiator.\n\nChange-Id: I0b8b5930c9e8f3c049b82e92c5fb7b05d6ae9291\nCloses-Bug: #1839384\n(cherry picked from commit e27d83f4d03a89d982aebbb3fb8a8d1adafbdfc5)\n(cherry picked from commit 87b933db778ce5751a75e1dd10ce3a442f9e6902)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9c6d4a7f7ac4bab4f30dc6a323e7eed9863a930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9c6d4a7f7ac4bab4f30dc6a323e7eed9863a930"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
