)]}'
{"id":"openstack%2Fcinder~727602","triplet_id":"openstack%2Fcinder~stable%2Frocky~Ib86729488ebfb0aea5d5d4f815a64e00258040e7","project":"openstack/cinder","branch":"stable/rocky","topic":"bug/1794646","hashtags":[],"change_id":"Ib86729488ebfb0aea5d5d4f815a64e00258040e7","subject":"VNX: delete the LUN from VNX backend","status":"MERGED","created":"2020-05-13 02:39:16.000000000","updated":"2020-08-03 13:07:36.000000000","submitted":"2020-08-03 13:05:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727602-1596459955034-bf775c37","meta_rev_id":"c4e4357d0dafa0f56c77982cc2ac4fc69999a087","_number":727602,"virtual_id_number":727602,"owner":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-19 03:14:22.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-08-03 03:17:30.000000000","_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"date":"2020-07-06 15:49:52.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"date":"2020-07-06 17:52:44.000000000","_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2020-07-06 18:05:34.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-19 10:07:04.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-06-22 05:47:51.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-06 15:20:08.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 13:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-06 17:23:50.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-06 15:52:17.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-03 06:03:39.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-03 07:33:33.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-06-19 14:28:52.000000000","_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"date":"2020-08-03 07:29:00.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-07-06 18:29:40.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-06 18:04:33.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-06-27 15:59:57.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-06 15:36:35.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 02:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":1,"date":"2020-06-19 02:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":1,"date":"2020-06-19 13:42:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"value":1,"date":"2020-06-19 02:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 02:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-03 11:40:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 21:58:58.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-06-09 08:02:36.000000000","updated_by":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"reviewer":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"state":"REVIEWER"},{"updated":"2020-06-15 19:09:48.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-06-19 03:14:22.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-19 10:07:04.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-06-19 13:42:53.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-06-22 05:47:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-27 15:59:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 15:20:08.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-07-06 15:36:35.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 15:49:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-07-06 15:52:17.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 17:23:50.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 17:52:44.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-07-06 18:04:33.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-07-06 18:05:34.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 18:29:40.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-03 03:17:30.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2020-08-03 06:03:39.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-03 07:29:00.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-08-03 07:33:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-08-03 11:40:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-03 13:05:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffa1c32b6ce45f89968ac6a3661473917e60486e","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-05-13 02:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b516a43474e23da7f20d19e992845ec2b58a2395","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-05-13 02:48:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/727602/1/iscsi/ : FAILURE in 8m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/727602/1/smb/ : FAILURE in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a5d0473e0a02a10d7d48b485969451b5cf6faccd","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-13 03:01:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/02/727602/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/0508ad1 : FAILURE in 21m 37s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/02/727602/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6a54976 : FAILURE in 22m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"33b65a10d10a7ce30ee7d5f2e470333297789b8a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-13 03:19:07.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/dce695e85f284c088cc3c5ce84cd2d87 : POST_FAILURE in 19m 09s\n- devstack-plugin-ceph-dsvm-bionic finger://zuul-v3-executor.novalocal/ff08b416514b48d1997f9c30458d42b7 : POST_FAILURE in 17m 46s","accounts_in_message":[],"_revision_number":1},{"id":"df150dbadd24fd64f1d1ac7fae0b8aa458aa8957","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-05-13 03:19:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/02/727602/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ade0661/ade0661.html : FAILURE in 38m 18s","accounts_in_message":[],"_revision_number":1},{"id":"947c0cb8fe86579109ce544312de5d7272ee8423","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-13 03:36:40.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/02/727602/1/check/dsvm-tempest-infinibox-fc/57fdc67 : FAILURE in 29m 09s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/02/727602/1/check/dsvm-tempest-infinibox-iscsi/f67946a : FAILURE in 22m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a0f5e30f520ed0380c9897dcd694098c12166729","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-05-13 03:43:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/44e5ed4b6bc2cd34eebbaab6cdf8058e2ff49118/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"37b86197114c6e0b930f5ca8ba03629478af76a4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-05-13 03:50:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/02/727602/1/check/dsvm-tempest-cisco-zonemanager-job/2eb89e7/ : FAILURE in 34m 02s","accounts_in_message":[],"_revision_number":1},{"id":"777149e4a2598b15fc1ac97175bf3730c9cf5ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 06:27:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2130c67c43845d2a3401c37f75f2597 : SUCCESS in 9m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36edf0ef14bb4873886fcb201c56a1c1 : SUCCESS in 10m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/21dcbb8e02124266a7c04796703df222 : SUCCESS in 9m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c7a8adb316204c159e7c996a95ae6295 : SUCCESS in 10m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7a39cf0270a940d2856f7ce19100d248 : SUCCESS in 1h 26m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7e491c645518433a8b4288cb067b3cc9 : SUCCESS in 51m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd85ef4172b341cebf85f0cc3a7e3e6b : SUCCESS in 1h 17m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80794e151c97486fb4f39ba627a528c5 : SUCCESS in 25m 04s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d98ea10994cf4d86abd3e93b1e2ad433 : SUCCESS in 2h 01m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/953983a927b34e72add0c89fd0ffa319 : SUCCESS in 8m 59s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/e2ccaefcd7b54e68ba79d5a8a47b9a31 : SUCCESS in 1h 09m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/75bed33f49834a56818e53edf74f0ef2 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b4f3091204546ddb93fbd9b3a83fd01 : SUCCESS in 8m 55s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/787323b8becd4585ab4a48d8a68becfb : SUCCESS in 6m 47s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/ac61fc3ae3d5412fb0efe2df8e19986d : SUCCESS in 10m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae90e8aa23404f9dabcc8707e790c728 : FAILURE in 17m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b12359efe3d4e52a350598b95d6c00e : FAILURE in 26m 04s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/999369f097af4fc38d97898aad8bad91 : SUCCESS in 1h 08m 55s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c0792aca44154f51bf78519f21ad9291 : SUCCESS in 5m 24s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5808cd035b124e379af2e22c982b6940 : SUCCESS in 1h 04m 15s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2370dbef2bed463dbcf9629924311d12 : SUCCESS in 1h 14m 48s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/88d0efbaa2ed4b42bc0dcd06dc629370 : SUCCESS in 1h 12m 08s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/98a499aa7e4041e38e5f32ef6ec88de1 : TIMED_OUT in 3h 04m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5bf28a5585ce6010970bd5b92c231e0149ffc15","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-05-13 07:39:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/02/727602/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2274f29/2274f29.html : FAILURE in 36m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8a5894cc93d59b564850f0c715d5639ff2432349","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-05-13 10:13:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/02/727602/1/upstream-solidfire-check/cinder-solidfire-iSCSI/2e204e9/ : SUCCESS in 1h 01m 33s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"aa016be5c809a92502e3f292dd278b0ad7e26fef","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-13 16:41:28.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5b9ec9e232feb8e290d2fec947933e241d007614","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-05-13 17:42:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/02/727602/1/20200513235946/NEC-FC : SUCCESS in 2h 40m 13s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/02/727602/1/20200513235946/NEC-ISCSI : SUCCESS in 2h 42m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b4b2dbd99ab1dcb381e13eb55b3358c484be71a0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-13 17:49:45.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"829e50cdf0a8751b3cc6e53d3381299cb4230ad9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-13 19:01:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/02/727602/1/upstream-check/cinder-cDOT-iSCSI/4419166/ : SUCCESS in 46m 09s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/02/727602/1/upstream-check/cinder-cDOT-iSCSI-IPv6/2bffad5/ : SUCCESS in 1h 44m 47s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/02/727602/1/upstream-check/cinder-cDOT-NFS/24b9be4/ : SUCCESS in 47m 50s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/02/727602/1/upstream-check/cinder-cDOT-NFS-IPv6/a95fb59/ : FAILURE in 3h 58m 07s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/02/727602/1/upstream-check/cinder-cDOT-FCP/337e07f/ : FAILURE in 1h 20m 32s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ac0dbc4ae639e46930861dde26646daea1a4a53c","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-06-04 21:58:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dc9268665dccb257e4b7e62703f67fe380d3c76f","author":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"date":"2020-06-09 08:02:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db6b4cbd01af5c3a358c02e9a7862fa5221df7d8","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-15 19:09:48.000000000","message":"Patch Set 1: Code-Review+2\n\nThis looks like an appropriate backport.","accounts_in_message":[],"_revision_number":1},{"id":"4f65e9859964c351f72bd97f97c0ea56293ab54e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-15 19:12:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged to stein, good to go for rocky.","accounts_in_message":[],"_revision_number":1},{"id":"28b2dd1b7a2d3505fb5ed6bb1f04a2ca3f330e8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:12:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"027bd05fde5fd5cfd70a4484e9195cd60cd5f3c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 20:34:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b7d4a746d024c0f8377be51a13355d0 : SUCCESS in 13m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a934ac138e74277a5e27cf78c36d498 : SUCCESS in 17m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/de7e76f0a6e14735b2aabcf75cddc550 : SUCCESS in 14m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fc0b7436c28a4f008fd75390feff6318 : RETRY_LIMIT in 5m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/091d438700a14864a1dadf028218440d : FAILURE in 16m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0a720a3798a447c9953dab1d51be8a70 : FAILURE in 24m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3314aa853bcd48b3a0e83d5672d2efc3 : FAILURE in 15m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b4e8a9a6484752b8799d83bd3e5cbe : SUCCESS in 10m 00s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/1a22558cf08843eca2813ee1b8ac2922 : FAILURE in 14m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/01ab63f83f6f425a841f41e9c64b0a5e : SUCCESS in 10m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0565e92e5dca40a7b73d4997d803d185 : SUCCESS in 12m 51s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/846c68460e4342f8a2c6a5993295c173 : FAILURE in 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"67a5e9f57e9adc0325a6adcafa68966185bab214","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-06-15 20:49:19.000000000","message":"Patch Set 1:\n\nNeed Dell EMC 3rd party CI results","accounts_in_message":[],"_revision_number":1},{"id":"84ef08515af2c7275402d785cb30628f8dcee8c5","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-06-16 05:56:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9562da6b903722d3ed5a0f107c91f4c0bc1c6292","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-16 06:04:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/727602/1/iscsi/ : FAILURE in 7m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/727602/1/smb/ : FAILURE in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"facee6c6c160797b987002de00b59291a5fbc92b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-16 06:14:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/02/727602/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/65a4584 : FAILURE in 17m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/02/727602/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/461d6af : FAILURE in 17m 26s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"4992c084706e571bef9b4455468d622ae6f3833b","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-16 06:17:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1210/index.html : FAILURE in 17m 28s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1359/index.html : FAILURE in 20m 32s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1385/index.html : FAILURE in 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4466d35b066526874a11a130dd6090769ee0c285","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-16 06:26:02.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/1/check/tempest-dsvm-full-bionic/e3278c8/ : POST_FAILURE in 22m 06s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/1/check/devstack-plugin-ceph-dsvm-bionic/717b202/ : POST_FAILURE in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d1b930e676236edc207ce84d5bc22263f9a702a6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-16 06:27:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/02/727602/1/check/dsvm-tempest-cisco-zonemanager-job/a36007f/ : FAILURE in 25m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0b996c057f1699130d60384c66e0739a60f5d101","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-16 06:29:44.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8684fef173c3a9690b59396a39a59e77d63a81c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 07:06:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe5815221c543ef915e2c8eecef0bc8 : SUCCESS in 10m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/205d3309a8464d888f6b8b78d70c5387 : SUCCESS in 15m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7162e82298ac4595b0433a5cbc1c0bb3 : SUCCESS in 15m 53s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6cb39c9cacd047168853231c15b369bc : RETRY_LIMIT in 5m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b737ae8919f048dea04f4791b625d218 : FAILURE in 10m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/25fef0cd1eda4eb2a48c2a12d57f66eb : FAILURE in 13m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8498d1d955f45bf87d4e59e3a8eff54 : FAILURE in 15m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df28fa4648824f1e99f56e8aeaf5be3f : SUCCESS in 7m 33s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b4d96fc2723e4c96b49f1edc1ed4ffd5 : FAILURE in 28m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30da67c678c842f5bfaacf443e13b683 : SUCCESS in 11m 28s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/42ae62fa77df48fdad2a952f29cf1b7e : FAILURE in 12m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/38d889ed9dde4b0ba04585222cbd592c : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d79955b2fae644bdbf0f660bcd9651a8 : SUCCESS in 15m 06s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/643e19aeeebf4e3d9235b3f3f8af194d : SUCCESS in 5m 36s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/b3e4958fdb8f46aa8f3c3118c21c599a : SUCCESS in 10m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/661fb93755084ae4b61a63b93732043a : FAILURE in 14m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/492dd1be6a6342e0bf4d6e1b58732e4e : SUCCESS in 31m 37s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2048af3dfcb7423194a9b83c0a2b6988 : FAILURE in 19m 42s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4d56e3217fde4f7db165c7dd5d236079 : SUCCESS in 8m 11s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/32265423b9c74541a5346666e9c27a86 : FAILURE in 28m 19s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cdb65edc6a1d4d1ab4b4ef33e49bcace : FAILURE in 14m 13s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/406cda126c4641d78452a4bca80ffff3 : FAILURE in 19m 14s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/dcbea71968d54f9593c18b36836df426 : FAILURE in 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00ad407c8759bfc48288a67b209d85368d493b18","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-06-16 23:18:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"709f353cbcc63750c89326a16e486d83a67e7157","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-16 23:27:15.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/727602/1/iscsi/ : FAILURE in 8m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/727602/1/smb/ : FAILURE in 7m 45s","accounts_in_message":[],"_revision_number":1},{"id":"692e381306cd8e210212c879abe88f99461e8984","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-16 23:50:48.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"aba443cc666da27878d61ec044e3d0468b275b63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:51:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e74c88d0542455e9d3429946ab21e2c : SUCCESS in 9m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6369e9e74be410fa37afbcf3e248faf : SUCCESS in 10m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f2c4c9f1d6774a7c818dde4327738153 : SUCCESS in 11m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d47b9d48436340f88bce463c7f1dc453 : SUCCESS in 8m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7a6943afe6bb4d6f8c5a81daaea5922a : FAILURE in 13m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bbae3b4ad39643efb596271a73b5487d : FAILURE in 16m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5aae31ee1f774c66b46c239f6cf443d6 : FAILURE in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce0c7f444a2141b3bcadebde64067282 : SUCCESS in 8m 26s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0fd58c0e8f864f3d86f74231f3fd4475 : FAILURE in 15m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d68a652d799d43928a29584eba9b622e : SUCCESS in 8m 55s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/ec6cf2d962ae4a7faa40e1bda4974384 : FAILURE in 15m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/9d21125b320249a38573cd769a29e62b : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78c1f8bf59a54d9e8d09c2c1a66ebaba : SUCCESS in 10m 43s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa18fb3197e540278acabf92d33c3bae : SUCCESS in 5m 52s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/46864ad75ffa4a0aa4b49b5c008a57d4 : SUCCESS in 10m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/19518ddc5f7344e3aec0c54fd856acd6 : FAILURE in 13m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3bc71bbe48904204b6b6fc4d623416cd : FAILURE in 28m 13s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/76351dac44c04a21b9936340bef863b0 : FAILURE in 15m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9a5d3a6413704a29a8051718acabaa34 : SUCCESS in 6m 50s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/42200408ff8641bf84417f7692162e0d : FAILURE in 17m 58s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3fe514ea5afb42a6a02c996573ea7841 : FAILURE in 12m 01s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/2b69eb84dc50443680e70e337ec8721d : FAILURE in 15m 27s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/d24b88f8b04f4e0e9113e55a6d88f407 : FAILURE in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9daff14a7b4c113757bb59ad2e8719713aa45c62","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-16 23:52:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/02/727602/1/check/dsvm-tempest-cisco-zonemanager-job/980fbf5/ : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d5197b31ee9887b1b953b8beb424829aa6660bbd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-16 23:56:11.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/1/check/tempest-dsvm-full-bionic/f127ebf/ : POST_FAILURE in 25m 12s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/1/check/devstack-plugin-ceph-dsvm-bionic/a70552b/ : POST_FAILURE in 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0c66a3d9cc25ae70d5bfc1eaf24f868afe36ed75","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-17 00:13:13.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/02/727602/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/759342d : FAILURE in 20m 26s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/02/727602/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/762412d : FAILURE in 17m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"678f894b6318d2f908c8a9ebafd76b8d00ed638a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-17 00:30:12.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2736347bcc5273d769d8a2f5809ae5c79fa55f30","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-06-18 07:54:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22bb41a3638eacc545dcf72191a55f94c538c1bf","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-18 08:02:17.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/727602/1/iscsi/ : FAILURE in 7m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/727602/1/smb/ : FAILURE in 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ccb3564cf087417fd7f55a8d3799882c4c1aa2b7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-18 08:24:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/02/727602/1/check/dsvm-tempest-cisco-zonemanager-job/144b277/ : FAILURE in 25m 39s","accounts_in_message":[],"_revision_number":1},{"id":"83c76f559e446e4fd6379a5d8889245ed34ca3f7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-18 08:28:17.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a7dcdc5b39fb693e624113e3daef7ab135ed1581","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-18 09:18:22.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/1/check/tempest-dsvm-full-bionic/a928c11/ : FAILURE in 21m 59s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/1/check/devstack-plugin-ceph-dsvm-bionic/f1a59cf/ : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a9ba89ec124d283dab5d8c97ee23dda73afc8c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:20:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840a9514fe0e4c42bf040ab7313424e0 : SUCCESS in 10m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62cd0fd1767e46fbab9936588014f5bd : SUCCESS in 10m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b845667d7d9441e3b50b0630a31b8dfa : SUCCESS in 10m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d66eb0ea6ea54d80bc17e932f2c542da : SUCCESS in 12m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a02e320325654897a7ff86b390d4d646 : FAILURE in 13m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/da5e02e30be0434ebd5bc81d12410f9f : FAILURE in 18m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7fb26765ef54651b818eb373c1f8915 : FAILURE in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b2fd1facee6412da24c70bc62871aa2 : SUCCESS in 8m 15s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b85a251b366942fea21acc43beafdc46 : FAILURE in 13m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ab201a2d31b4c0c83c28275c8da0aa2 : SUCCESS in 11m 49s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/bd2f59e90aca4fb6b72a9770f1b4162e : FAILURE in 16m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d3a6eedaf27d4c36a3f03da4506f11fe : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56eb56ea78e249f8b3adf57b4477aac1 : SUCCESS in 11m 53s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65422fab79ae4dd4868411fffbcca341 : SUCCESS in 6m 48s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/3cafe24fdc154fd1892634f27091e120 : SUCCESS in 10m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/910d7ef082134f51aeaf8e2d71bbf77f : FAILURE in 8m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33d54982a6514d97ad69197844ba99d8 : SUCCESS in 17m 08s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e8bf32cabf1d467d86671bbef5917f2b : FAILURE in 9m 04s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8f8c59c46d8647248735c4e36de37f0a : SUCCESS in 5m 35s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/14c406a00c5046eeb5df61a5ec4b3405 : FAILURE in 17m 12s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9c61138e8e841358affedf06e9091b5 : FAILURE in 16m 58s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/1ee9dc9f1cd4408ea566f16561555931 : FAILURE in 17m 28s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/e508bb24af9d4fbd935ba00f28b072a4 : FAILURE in 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0deb3a5936a4d9b6355bb90e75a4d68d6bcafa76","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-18 09:51:55.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d91155861112127210a3fdfbc7fbce75b175907a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-18 10:01:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/02/727602/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/1c3ff16 : FAILURE in 17m 27s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/02/727602/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/4f60a2f : FAILURE in 18m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"689e416490b8fedd9d4df31838a919776abc7ecb","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-06-19 02:21:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"44843f899bc3198fe15161103e95c51a16d49928","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-19 02:29:37.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/727602/1/iscsi/ : FAILURE in 7m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/727602/1/smb/ : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8cc5ba46db61d2de023adaab44010614e1466958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 02:56:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3f693442073431d940515aa2aeb71ea : SUCCESS in 7m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7298bd970924b9385c62d1f00952b72 : SUCCESS in 10m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/70ac1a1e1d054d6782df9a2453045d47 : SUCCESS in 9m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f755d0614f124511980ea39348fe906b : SUCCESS in 8m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b7a01c9497a84320a1e2eb2c0be47333 : FAILURE in 9m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b035986268f54e08bd99789ac69d98bb : FAILURE in 16m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/365fd3278f3d4b65aa563474f64b59eb : FAILURE in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2322741805bf40569302c7813e02f0f9 : SUCCESS in 6m 33s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/252cf7c6f94e4761b1cbde7ed5435acd : FAILURE in 12m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c20b2982888e441d8505881e133497ed : SUCCESS in 9m 08s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/2402f96113b546b4a20dd0678fc393ce : FAILURE in 12m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c20aef70c3514d329c08bc7704ff69fe : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c489438379a4d40ba219c00fa2afeb6 : SUCCESS in 10m 49s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/549bff66f83a4821a5975fc986253cd8 : SUCCESS in 5m 59s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/b720b374b3ea4130a7ec33fa0bb7ba62 : SUCCESS in 9m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/64afc1be148f4207883977d5487a5fb5 : FAILURE in 12m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a93b716aa06462a950a853f453b2508 : SUCCESS in 34m 02s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ba27d42a82646ca9f62c8832b647b86 : FAILURE in 14m 27s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/217e508f4b9547f296d786663a760d54 : SUCCESS in 7m 16s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6171fa5f5b824cd9b8ff25ecb2a7877f : FAILURE in 14m 19s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e7af69f555834553b2fbe43915f5091a : FAILURE in 14m 36s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/20d5e66c83364e23b745ddc85d29c597 : FAILURE in 16m 08s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/8a172f6b14d6487e880c3612b2b2e788 : FAILURE in 12m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e334f133d9267272779fadd574cfe3a508d41bf0","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-06-19 02:59:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c346ee636f4021bf73d0dda2254e4812c6401712","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 03:02:26.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6a70fa4720d687cf6c5601f4a2aff25ff437c332","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-19 03:08:04.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/727602/2/iscsi/ : FAILURE in 8m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/727602/2/smb/ : FAILURE in 7m 39s","accounts_in_message":[],"_revision_number":2},{"id":"bce6f0c7e93143c53889cffb94352c30d3b627e7","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-19 03:14:22.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/6e2e278f3c97ee14d7b18272cee5bfea1c231768/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1bbacf9a515b7dae2b9fa58c1b182525426fd595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 03:23:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69eddc8bebe243de966c90c7a768f8e8 : SUCCESS in 10m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46678fa5f67c4b36a2fc699d8948e789 : SUCCESS in 10m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b4b2806a20ee4f75bdf27c7562fd2115 : SUCCESS in 10m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b380346658bd4075874a1d91a328b8b9 : SUCCESS in 9m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c58025142837474b9e57f35141e92a2c : FAILURE in 14m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1df1d7b52189402f8216bd55375441fb : FAILURE in 11m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc728ab73e7a4a1babd8e8685c747a01 : FAILURE in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4dd88149864b2bb36351494ac11d9d : SUCCESS in 7m 24s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/610d3c3accd5487ab5ef05d5895fd469 : FAILURE in 16m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af22b06eb22a469197373c1d5dc028e7 : SUCCESS in 9m 11s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/964fc977b6bb42daafe28f5137013a1a : FAILURE in 16m 33s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1f6618948c8c405fa49b78858969c784 : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d54cf53d731a4dcd82ff68ebbcb72c80 : SUCCESS in 9m 42s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b83a720a67514b68883dab6d297ceb5c : SUCCESS in 5m 45s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/f38f62a30e894125909f14bb5fc0e920 : SUCCESS in 9m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/487223f98c414f668a13649ec2c0b124 : FAILURE in 12m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d6f6ccde84d4437da9b1fbe93e551121 : SUCCESS in 22m 46s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/da1b9b13381c4a2097f3efd793835744 : FAILURE in 14m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6796fcd6c1f14689b14c8958afcaab2d : SUCCESS in 6m 18s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9160567802bd4ea2a9ca138aaebfae99 : FAILURE in 18m 36s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ea0f1300e1b94130994e6d203167bc3b : FAILURE in 14m 54s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/980596572ae84994bc8ccc40ca82ca79 : FAILURE in 15m 36s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/43defb1c70db45e7af7f0d910c0229a6 : FAILURE in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b681755fd7610c48335073ecd59ea33af1765053","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 03:44:24.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"dedaa5aeb48492024ba106b71a3e6d70168a20ee","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-19 03:58:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/02/727602/2/check/dsvm-tempest-cisco-zonemanager-job/9b1e4d1/ : FAILURE in 25m 16s","accounts_in_message":[],"_revision_number":2},{"id":"6fe2634444f7fbc87295cf2c0592c721922eca79","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 04:17:50.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3b98b6b5a0c9466a2c3a45a7f0cf8dba1d3b046d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-19 05:12:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/02/727602/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/baeefb8 : FAILURE in 17m 28s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/02/727602/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b55ff0d : FAILURE in 17m 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"ec0bc8bc150e8d378fbb4e9a8b3305c4390f65d4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-19 05:43:29.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/2/check/tempest-dsvm-full-bionic/aa6f2cc/ : FAILURE in 23m 32s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/2/check/devstack-plugin-ceph-dsvm-bionic/e681e76/ : FAILURE in 24m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2e694cc88d51515b07f61eff126e78e72e81c4bd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 06:49:05.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"7c5531bf110d999563b37f3a44046d992bd9739f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-19 10:07:04.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/02/727602/2/check/dsvm-tempest-infinibox-fc/d31ba1e : FAILURE in 28m 39s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/02/727602/2/check/dsvm-tempest-infinibox-iscsi/5bcd3c3 : FAILURE in 21m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5ed01fe02b4ced76b3cbab99d735cd6c1eeed9b5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 10:40:11.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"82adf2f6ae73d8dadfd201a2ff3d220a81fa50aa","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-19 13:42:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c4471ff29b569a8545a92aeb027651381704bdac","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-06-19 14:28:52.000000000","message":"Patch Set 2:\n\nThe zuul check keeps failing due to this issue https://bugs.launchpad.net/devstack/+bug/1883468.","accounts_in_message":[],"_revision_number":2},{"id":"be57bb4fbef42605477c7ef1874d66a06cbd46ef","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-20 09:57:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1239/index.html : FAILURE in 14m 03s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1386/index.html : FAILURE in 16m 37s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1411/index.html : FAILURE in 14m 42s","accounts_in_message":[],"_revision_number":2},{"id":"efa9d7f6f1066ef3e8f210a60d4e32730287bfcb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-22 05:47:51.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"95b98141193e8f230db662481c49123aef45c335","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-06-27 15:59:57.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/02/727602/2/20200628004431/NEC-ISCSI : FAILURE in 14m 11s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/02/727602/2/20200628004431/NEC-FC : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"e49d8fb67d3a24b0f697fbfd6927a94eca1e5efc","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-27 16:34:51.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ef62dda642dd787d14ee58649b5cef20780beb93","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-01 15:53:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6aa5af44b67b4b841c60006a6ca66f392d216440","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-01 16:03:47.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/727602/2/iscsi/ : FAILURE in 9m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/727602/2/smb/ : FAILURE in 9m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5ab2c50c50223b6a81e252f826eb7393bbf96d6a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-01 16:45:44.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/02/727602/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/904d975 : FAILURE in 22m 50s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/02/727602/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/afc273e : FAILURE in 22m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"deabf9d01f4e21e9932404e8aedf059116323c38","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 16:50:14.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"fec652d32adb8e1cef5d626afd8319b338145f82","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-01 16:53:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/02/727602/2/check/dsvm-tempest-cisco-zonemanager-job/8a9290f/ : FAILURE in 33m 11s","accounts_in_message":[],"_revision_number":2},{"id":"667e7152f743f72b32f80fa7300398668d462d74","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-01 16:59:55.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/2/check/tempest-dsvm-full-bionic/5e0a81e/ : FAILURE in 25m 51s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/2/check/devstack-plugin-ceph-dsvm-bionic/9245697/ : FAILURE in 34m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2da3eb7096134bcc37ebe9c2fac4ef0787a973fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:38:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72239f6450c44f62b92dbc6c7bad7b94 : SUCCESS in 11m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9ea681dd9314bf4a70f42c29b293849 : SUCCESS in 11m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/de8175a0223e4a16b4c3652bc0f894c6 : SUCCESS in 16m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8de65b2de9d646beb3159b691984f3fd : SUCCESS in 10m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a8694679e638476bb0896eb90c0d4a3d : SUCCESS in 1h 36m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1234e39b2b634cf79811507a0b6dd0e8 : FAILURE in 8m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1406086c87a4657aacaf49de9708e4f : SUCCESS in 1h 31m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bf57e0cd6924aeda20033f88ed4114f : SUCCESS in 8m 06s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c006276951ba45eab12a607ee1b786cb : SUCCESS in 1h 06m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63cd48ca2c1b499f9b58179f7369c0b1 : SUCCESS in 11m 49s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/96133784146d4c4cb75b290f06bddf9b : SUCCESS in 42m 53s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/12b6cd08f0a64b769a69feeccced2982 : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2336b8cef25d481fb4daaa8a5a245c46 : SUCCESS in 11m 46s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b1eb752660f452fa69d39eed60e2829 : SUCCESS in 7m 28s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/675fbe47dc8e4d6791cb7fc8c23bad1e : SUCCESS in 10m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e42858f8f6d647b88248f809b7ad8015 : FAILURE in 22m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e9a093cac114745a9f4e21dff69dfbb : FAILURE in 19m 32s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d01fac79b5ed40b39f843f80ef19ec7d : POST_FAILURE in 1h 36m 54s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da8ca2fd12cb48468690f33463f24266 : SUCCESS in 8m 00s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/64c902cddf8e49b39eaa65f897c58f97 : FAILURE in 21m 12s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c49afaeb61d448bf9c8dd9404c52cc1c : SUCCESS in 54m 42s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/1d6f48736e594853982530263820b33f : SUCCESS in 1h 26m 27s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/ef83689a2b634569820b9665f29fbbe3 : FAILURE in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7324fed3c45a6c247864e37af8f85489eef3b226","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 18:06:36.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"afcd66378a9fe72679e506a1beaeb3d75412e50f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-06 15:11:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"729e8a37e17b4652d74393ff027fd81c2c814518","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-06 15:20:08.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/727602/2/iscsi/ : FAILURE in 7m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/727602/2/smb/ : FAILURE in 7m 48s","accounts_in_message":[],"_revision_number":2},{"id":"570ee62934ae661929fb42a53a20a72de459184e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-06 15:36:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/02/727602/2/check/dsvm-tempest-zadara-driver/4ab1eeb : FAILURE in 24m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"918205083eceb6e14afa5f2f23607f4788858d34","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-06 15:46:04.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"70bf152b09a9277c305d6963535759982cb746d6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-06 15:49:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/02/727602/2/check/dsvm-tempest-cisco-zonemanager-job/ff547f1/ : FAILURE in 32m 54s","accounts_in_message":[],"_revision_number":2},{"id":"096025b02650ab357bc185041375dbf6db677a19","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-06 15:52:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/02/727602/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/a26769f : FAILURE in 22m 42s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/02/727602/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/feebb70 : FAILURE in 27m 41s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"2be95d8ea1e6efa8f7f726622364eeea8b334f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:20:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06580914fd544b478f19e8120ff486ec : SUCCESS in 10m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fcf89ec3898846e28c6b9f5e8387309b : SUCCESS in 11m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/30911236fd964ff8b5f737a91e487d23 : SUCCESS in 12m 22s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9fc562457fd346c4b29b64e842e54885 : SUCCESS in 9m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d7a387655da245cead15419203003aa9 : SUCCESS in 1h 17m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8850976420d8460aace05f1e5c2c2bd0 : SUCCESS in 51m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bc61e3f1b524302bf7fa6672886a09c : SUCCESS in 1h 15m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f5d3a4950fd409184ec463a730b0caf : SUCCESS in 6m 35s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5c64be578dad47258c6d0b2e06f5dd6d : SUCCESS in 1h 39m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c0099b6a8be41e2bd5971133bfcd677 : SUCCESS in 10m 47s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/fef608e6f23944689b2dad97bb0efbba : SUCCESS in 58m 31s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/11a71b5e94084ad48233c76eb2eef175 : SUCCESS in 13m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46c6efd59b804bf78393abe36370f206 : SUCCESS in 10m 14s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9081fe09c1541dabad9e397e99a4587 : SUCCESS in 9m 23s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/63993b066ee944b8acc26a987a87196b : SUCCESS in 8m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0c6086fa244e45a68e2d487854628d40 : FAILURE in 20m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11ce5b5763f040c6b03b650b8a30306c : FAILURE in 17m 44s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a912c90957cc4dae9e56b004b53484da : FAILURE in 17m 27s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f289efcd4c53484497a245bad3f14de7 : SUCCESS in 9m 15s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce27f9dc1ed640e4a3410fc0eb76ce69 : SUCCESS in 1h 04m 15s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/42bdf3c3561747c0b636b5e722839083 : SUCCESS in 1h 12m 09s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/5627b83478d441818d16d5b937c4f6ae : SUCCESS in 1h 10m 49s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/151d3e024d884e429af281966fa2b1c7 : SUCCESS in 1h 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4262c15827d5c8875518922ec27affe609f1b384","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-06 17:23:50.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/02/727602/2/check/synology-dsm-cinder-iscsi/debf8c3 : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"3abb3a72aff6a5e0acd07d8197d20286e1418076","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-07-06 17:52:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1300/index.html : SUCCESS in 1h 09m 33s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1447/index.html : SUCCESS in 1h 33m 27s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1472/index.html : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c04b8198132c64e3020b97a035ec08d0c56d61e3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-06 18:00:25.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"23f958695df3e4dd23ee2be988ef67855fab4263","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-06 18:04:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/02/727602/2/upstream-solidfire-check/cinder-solidfire-iSCSI/098e677/ : FAILURE in 43m 16s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"ff97485e3076558f4fe6a1f1fb8393ecb91d07d7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-06 18:05:34.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/2/check/tempest-dsvm-full-bionic/a32277e/ : FAILURE in 24m 58s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/02/727602/2/check/devstack-plugin-ceph-dsvm-bionic/bdb589b/ : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2ab129ab810288d55ef2ccf3081b612971ee428d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-06 18:29:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/02/727602/2/upstream-check/cinder-cDOT-iSCSI/9fcc6dd/ : FAILURE in 1h 08m 42s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/02/727602/2/upstream-check/cinder-cDOT-iSCSI-IPv6/8a7ae24/ : FAILURE in 55m 34s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/02/727602/2/upstream-check/cinder-cDOT-NFS/8651371/ : FAILURE in 27m 12s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/02/727602/2/upstream-check/cinder-cDOT-NFS-IPv6/fc8dd05/ : FAILURE in 58m 01s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/02/727602/2/upstream-check/cinder-cDOT-FCP/53c3c64/ : FAILURE in 39m 52s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7f5c34a21a60457a05b70532739b4201467b643f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-06 18:39:56.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"9c32ce24fa94c5298394f020e8cfc967fba35141","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 18:28:06.000000000","message":"Patch Set 2:\n\nrun-DellEMC VNC CI","accounts_in_message":[],"_revision_number":2},{"id":"b86388d9a65065153db0f470dcc8de32d0bdcbdc","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-29 06:37:55.000000000","message":"Patch Set 2:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":2},{"id":"b52226579e9b11ba1ffa7817bd47641c16a038aa","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-29 06:59:00.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_iSCSI/c65e547/DellEMC_VNX_iSCSI/198 : FAILURE in 20m 45s\n- DellEMC_VNX_FC http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_FC/423c14c/DellEMC_VNX_FC/193 : FAILURE in 20m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"1892def660461c1b56da13db138e7d258ee2bdae","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-29 07:34:40.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8caa99b9308cded01a7f0e341b5821c4f8445bb6","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-31 08:14:42.000000000","message":"Patch Set 2:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":2},{"id":"b696ae4e66c0500d5f3605f42dc7ce7060016f23","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-31 08:33:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_iSCSI/7c170c5/DellEMC_VNX_iSCSI/218 : FAILURE in 18m 16s\n- DellEMC_VNX_FC http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_FC/b7e5c6e/DellEMC_VNX_FC/213 : FAILURE in 16m 50s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"cfe261092e01d0f18351a65873272ca89428bba7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-31 09:08:10.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3ead120495b944753f6a6d8588a439bbe8cd1a6b","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-31 11:41:36.000000000","message":"Patch Set 2:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":2},{"id":"2d9da5e2d94896893856b0dbd61b371289c4c436","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-31 12:06:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_iSCSI/f0c306e/DellEMC_VNX_iSCSI/220 : FAILURE in 22m 48s\n- DellEMC_VNX_FC http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_FC/c76e49a/DellEMC_VNX_FC/216 : ABORTED in 24m 49s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"efb4647e25c155f819804cbbd3401f9269820462","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-31 12:06:56.000000000","message":"Patch Set 2:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":2},{"id":"8c6c27050d4151b41322509e3c96f7fded5d6365","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-31 13:29:56.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2c300cdefe815942f89ce724a056f8ce3ac3c014","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-31 14:57:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_iSCSI/6316ecc/DellEMC_VNX_iSCSI/221 : FAILURE in 2h 31m 59s\n- DellEMC_VNX_FC http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_FC/eeaa700/DellEMC_VNX_FC/217 : FAILURE in 2h 50m 00s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"e87cd3e8cb8896133f026fa3e224be8884c3e288","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-31 16:45:36.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"095a326e768e124cfaf4ea86172913fc3a44400f","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-03 01:40:05.000000000","message":"Patch Set 2:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":2},{"id":"d0639b67e905cde4ecfe9bb18b43d23170c49a4c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-03 03:17:21.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_iSCSI/95b132f/DellEMC_VNX_iSCSI/227 : ABORTED in 1h 37m 00s\n- DellEMC_VNX_FC http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_FC/11e6b45/DellEMC_VNX_FC/223 : ABORTED in 1h 36m 58s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"b7cab2fe370d90546899c6a3a728cbc6bd868bcb","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-03 03:17:30.000000000","message":"Patch Set 2:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":2},{"id":"32fb7ed1360cc09fdbc73644f5e274ade7858da8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-03 04:55:34.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a53317b8a4163bd6034882df2befb5bf30a4c391","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-03 06:03:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_iSCSI/e1cd18f/DellEMC_VNX_iSCSI/228 : FAILURE in 2h 45m 53s\n- DellEMC_VNX_FC http://publiclogs.emc.com/02/727602/2/check/DellEMC_VNX_FC/cabb2f6/DellEMC_VNX_FC/224 : FAILURE in 1h 17m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"3e485d8043ea741edf1b59eca48b512939dc8f1a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-03 07:29:00.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/02/727602/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"47a1e866b14c6f015c877f0bfe11db43078e7a9d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-03 07:33:33.000000000","message":"Patch Set 2:\n\nAre you unsure about the solution or testing something or could we approve this? (This was already approved but we\u0027ve lost the +W during rebase)","accounts_in_message":[],"_revision_number":2},{"id":"ac2b30c16f861d3ed5dc9ffd07098fc8be00461b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-03 11:40:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1b7fd978a5be7d7df8035727b60389817d436112","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 11:40:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"463f2f0309d773c08709fd585b3ac1bfd64ff25f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 13:05:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b7ef55042c34c46bf030fbba8b563ba : SUCCESS in 8m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6150e2ef4679479a9d5b6ba1ba59a7bd : SUCCESS in 9m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d04fd05d52974501abc9920da863a26e : SUCCESS in 9m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/07ab168efaf7429491d5d04baeb2fb00 : SUCCESS in 9m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dbd073c53aa547bf83f18b2e0168445b : SUCCESS in 55m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/600271e741894075a3be7c59264906a0 : SUCCESS in 49m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c69434940351475dbef1402482997a6e : SUCCESS in 1h 19m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cd1abf5bbb64b9e8af7771d7061f310 : SUCCESS in 7m 35s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/f7cf8580ff2f4e0c888ec6c9d7b59cad : SUCCESS in 55m 52s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e11f0cd5ce564038bb5521dd2f42b7e8 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a20b1bd243f46e6926f7d7fbf49ce5f : SUCCESS in 9m 02s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/484ad3df76ba42e982f707fe6864578a : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ca859db52f1365541cc80600440716c74e22780a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 13:05:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2e4910a1264ea3c1b4bb67abbe84bd27570a345a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 13:07:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/038c331a1b354d3eaffe72f22e71a793 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e2e278f3c97ee14d7b18272cee5bfea1c231768","revisions":{"44e5ed4b6bc2cd34eebbaab6cdf8058e2ff49118":{"kind":"REWORK","_number":1,"created":"2020-05-13 02:39:16.000000000","uploader":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"ref":"refs/changes/02/727602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/02/727602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/727602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/727602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/727602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/02/727602/1"}}},"commit":{"parents":[{"commit":"79d3d4b45fb1fb905eae1665a6b4a742bfaf4ba5","subject":"Merge \"RBD: fix volume reference handling in clone logic\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79d3d4b45fb1fb905eae1665a6b4a742bfaf4ba5"}]}],"author":{"name":"Ryan Liang","email":"ryan.liang@dell.com","date":"2018-09-28 01:58:43.000000000","tz":480},"committer":{"name":"Ryan Liang","email":"ryan.liang@dell.com","date":"2020-05-13 02:38:38.000000000","tz":480},"subject":"VNX: delete the LUN from VNX backend","message":"VNX: delete the LUN from VNX backend\n\nAsync migration is used in the process of creating a volume from a\nsnapshot and an internal temp snapshot is created.\n\nBecause the temp snapshot isn\u0027t deleted, the LUN hosting the temp\nsnapshot cannot be deleted from the VNX storage.\n\nFor example, a new volume, V2, is created from snapshot S1 of volume V1.\nAn internal temp snapshot S2 is created from copying S1. V1 now has two\nsnapshots, S1 and S2. Although we delete V1, V2 and S1 from Cinder,\nS2 which is Cinder user-invisible isn\u0027t deleted and which causes V1 left\non VNX too.\n\nThe fix makes sure the snapshot S2 is deleted. Then the delay deletion\non V1 can be executed successfully.\n\nChange-Id: Ib86729488ebfb0aea5d5d4f815a64e00258040e7\nCloses-bug: #1794646\n(cherry picked from commit b3a89dc187ae4d71fb1fc27645cb4383209b0ef8)\n(cherry picked from commit a74945b0687c4aa69960777fb301984cd4d2fedd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44e5ed4b6bc2cd34eebbaab6cdf8058e2ff49118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44e5ed4b6bc2cd34eebbaab6cdf8058e2ff49118"}]},"branch":"refs/heads/stable/rocky"},"6e2e278f3c97ee14d7b18272cee5bfea1c231768":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-19 02:59:25.000000000","uploader":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"ref":"refs/changes/02/727602/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/02/727602/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/727602/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/727602/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/02/727602/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/02/727602/2"}}},"commit":{"parents":[{"commit":"c427484d75e276aa02dcd7586e2f8bed3780878c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c427484d75e276aa02dcd7586e2f8bed3780878c"}]}],"author":{"name":"Ryan Liang","email":"ryan.liang@dell.com","date":"2018-09-28 01:58:43.000000000","tz":480},"committer":{"name":"Ryan Liang","email":"ryan.liang@emc.com","date":"2020-06-19 02:59:25.000000000","tz":0},"subject":"VNX: delete the LUN from VNX backend","message":"VNX: delete the LUN from VNX backend\n\nAsync migration is used in the process of creating a volume from a\nsnapshot and an internal temp snapshot is created.\n\nBecause the temp snapshot isn\u0027t deleted, the LUN hosting the temp\nsnapshot cannot be deleted from the VNX storage.\n\nFor example, a new volume, V2, is created from snapshot S1 of volume V1.\nAn internal temp snapshot S2 is created from copying S1. V1 now has two\nsnapshots, S1 and S2. Although we delete V1, V2 and S1 from Cinder,\nS2 which is Cinder user-invisible isn\u0027t deleted and which causes V1 left\non VNX too.\n\nThe fix makes sure the snapshot S2 is deleted. Then the delay deletion\non V1 can be executed successfully.\n\nChange-Id: Ib86729488ebfb0aea5d5d4f815a64e00258040e7\nCloses-bug: #1794646\n(cherry picked from commit b3a89dc187ae4d71fb1fc27645cb4383209b0ef8)\n(cherry picked from commit a74945b0687c4aa69960777fb301984cd4d2fedd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2e278f3c97ee14d7b18272cee5bfea1c231768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2e278f3c97ee14d7b18272cee5bfea1c231768"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
