)]}'
{"id":"openstack%2Fcinder~729603","triplet_id":"openstack%2Fcinder~master~I0887f48c28741f434c6a48041c87e849e471bb94","project":"openstack/cinder","branch":"master","topic":"729603","hashtags":[],"change_id":"I0887f48c28741f434c6a48041c87e849e471bb94","subject":"[Unity]: delete lun with hosts accessed","status":"NEW","created":"2020-05-20 13:43:40.000000000","updated":"2021-09-17 15:22:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"287a63efd76b2397d8d21c55dae292887d6f475b","_number":729603,"virtual_id_number":729603,"owner":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-16 03:03:46.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-11-16 13:53:47.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"date":"2020-11-16 07:10:30.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"date":"2021-02-01 06:50:36.000000000","_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"date":"2020-11-16 08:43:43.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"date":"2020-11-18 11:26:37.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-16 07:55:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"date":"2020-11-16 06:36:47.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"date":"2020-11-17 07:26:17.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"date":"2020-11-16 05:49:49.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"date":"2020-11-16 08:35:37.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"date":"2020-11-16 09:40:51.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"date":"2020-11-16 09:37:43.000000000","_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"date":"2020-11-16 11:37:07.000000000","_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"date":"2020-11-16 04:02:42.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"date":"2020-11-16 07:09:54.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"date":"2020-11-17 17:32:58.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"date":"2020-11-16 09:32:32.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"date":"2020-11-16 11:41:55.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"date":"2020-11-17 06:41:10.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"date":"2020-11-16 06:33:51.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2020-11-16 13:16:55.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"date":"2020-11-16 04:08:50.000000000","_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"date":"2020-11-16 15:52:52.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":-1,"date":"2021-09-17 15:22:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-16 07:03:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":1,"date":"2021-02-18 09:52:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 02:34:57.000000000","updated_by":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"reviewer":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"state":"REVIEWER"},{"updated":"2020-05-21 03:24:19.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"REVIEWER"},{"updated":"2020-06-01 03:30:56.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-01 06:04:23.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-06-05 19:09:07.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2020-06-16 03:34:52.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-06-18 05:48:27.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-06-19 04:52:33.000000000","updated_by":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"reviewer":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"state":"REVIEWER"},{"updated":"2020-07-24 18:32:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-23 07:28:32.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-09-23 11:27:15.000000000","updated_by":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"reviewer":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"state":"REVIEWER"},{"updated":"2020-09-24 06:30:44.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-24 07:51:59.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-24 12:32:09.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-24 12:35:04.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 00:22:07.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-27 04:55:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-27 05:03:53.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2020-09-28 18:36:21.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-10-02 18:38:30.000000000","updated_by":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"REVIEWER"},{"updated":"2020-10-03 07:20:54.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-04 04:07:16.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 03:03:46.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 04:02:42.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 04:08:50.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 05:49:49.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 06:33:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 06:36:47.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 07:09:54.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-11-16 07:10:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-11-16 07:55:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 08:35:37.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 08:43:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 09:32:32.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 09:37:43.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"REVIEWER"},{"updated":"2020-11-16 09:40:51.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-11-16 11:37:07.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-11-16 11:41:55.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-11-16 13:16:55.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 13:53:47.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 15:52:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 06:41:10.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 07:26:17.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-11-17 17:32:58.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-11-18 11:26:37.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-03-16 07:03:17.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-09-17 15:22:31.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"9b1d2f795e6bab5d647536324aeb832d6ef44657","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-05-20 13:43:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f65e30b3778e7c54bd94faa11592fe172059fab8","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-05-20 13:51:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94391501c92a43da3d463286f783522a55ab6bb1","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-05-20 13:58:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/68c9b0ab03b2cf4be62f31714ddcf2a33309c44c/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0bbf7bca1743aa1c6454694c40483c894cc88a69","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-05-20 14:01:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/729603/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f37e760/f37e760.html : POST_FAILURE in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"17f4700a60129561f9d4aeffa7b09c20781798ca","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-05-20 14:28:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/03/729603/2/check/fujitsu-eternus-dx-iscsi/d6930be : FAILURE in 36m 04s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/03/729603/2/check/fujitsu-eternus-dx-fc/d6a2d7d : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6db98ba7953c5087a5f1ed511909a50e15e26750","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-05-20 14:35:16.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5ff9208add6f937b82699ed65d4ead4ae540fab9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-05-20 14:40:03.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/2/iscsi/ : FAILURE in 47m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/2/smb/ : FAILURE in 46m 25s","accounts_in_message":[],"_revision_number":2},{"id":"df0e2aa3d18268fe085211e91901fbe86cf670ec","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-05-20 14:43:58.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3181d2ad011c879f2736eae9f54b09de45e74f63","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-05-20 15:19:57.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/cba14e50e6981feef6e6b71410990f353bb70229/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e20886c523009165a3f387f35ef1020847610d9a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-20 15:49:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/2/check/DellEMC_Unity_FC/97368d1/DellEMC_Unity_FC/27 : SUCCESS in 1h 29m 20s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/2/check/DellEMC_Unity_iSCSI/a5eb13e/DellEMC_Unity_iSCSI/27 : FAILURE in 1h 57m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"18b2fa1cc5184b8dd342ee7bc5e82f5c5ed2122d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-05-20 16:08:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/2/check/dsvm-tempest-cisco-zonemanager-job/8e0f4ed/ : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":2},{"id":"66a3e3ae30cc3e4a27128486ff5f3d5fd1f06848","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-20 16:11:01.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-729603-1/ns5_nfs : SUCCESS in 2h 2m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-729603-1/ns5_iscsi : SUCCESS in 2h 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"98a4dc5c75e55d73d1979acfa534855fc5465f3c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-20 16:12:38.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-729603-2/ns5_nfs : SUCCESS in 2h 20m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-729603-2/ns5_iscsi : SUCCESS in 2h 2m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ef1e357d707f73529c0dd2e3ed12844d3bdc88d3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-20 16:38:32.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/03/729603/2/check/dsvm-tempest-infinibox-fc/4a76c4e : FAILURE in 1h 19m 59s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/03/729603/2/check/dsvm-tempest-infinibox-iscsi/c0cd124 : FAILURE in 1h 07m 03s","accounts_in_message":[],"_revision_number":2},{"id":"cbcbeafe894cbcbd795c184b12b34612cf139c60","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-05-20 16:41:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/729603/2/check/cinder-storpool-tempest/bd0e063/ : SUCCESS in 2h 30m 03s","accounts_in_message":[],"_revision_number":2},{"id":"38aacf032d1418bc623d5e2b5c6ae894145f8a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 16:47:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d2f0f972e4c421caf9372d95d0d1819 : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326a78a006994cde836dae866760a360 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6b3e52e129c4e2982d9d5c9b4a78218 : SUCCESS in 12m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50e04a2dd8f641b0a0971056610d0dc2 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ac2ff85da5424ca0d93c46b34db26f : SUCCESS in 14m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/2f017ea2ea334e5e9642e05c1dab19ad : SUCCESS in 58m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44a36735665b4548a82043277066d511 : SUCCESS in 1h 07m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b7566506fc74edbb12ed3228c960940 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f18f1e00befc434fbb1da7e83c808638 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/011ca4c7dbb244b1b6211820caa5e363 : SUCCESS in 10m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7d737df0beac4f31b4c4874fd2dae2f1 : SUCCESS in 41m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ad5ca5b0a7442ff844a74c659456852 : FAILURE in 7m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce65403808c14ceda128fc27a780f37d : SUCCESS in 1h 14m 50s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ff80a19473734cbf930d89808b8a4488 : SUCCESS in 1h 28m 09s\n- cinder-grenade-dsvm-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc2c6a99928842dc9ad5bf0275a3ee57 : SUCCESS in 1h 13m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bdc49f8cf0b04ace936f904469551202 : SUCCESS in 1h 27m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/48e17c169b944ee3a7453377b01247a6 : SUCCESS in 1h 22m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eafc5357ca344ace9d28ad3d1921d828 : SUCCESS in 2h 23m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2b02aefca56c447bb4a4fbe75f5b010b : SUCCESS in 1h 08m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f261b5dc45c14eb09a64f006040d65be : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7b3d151a4c3288f0796507f490431799f0b57d1e","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-05-20 16:52:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/03/729603/2/check/nimble-iscsi-driver-dsvm-volume/876080e : SUCCESS in 1h 40m 15s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/03/729603/2/check/nimble-fc-driver-dsvm-volume/b2bd416 : FAILURE in 1h 08m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a3702ba012eb4e2d0ac0ef63ba6c4a8d763f7fc1","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-05-20 17:40:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"893ca9b27928db7521b83cecbdeb57aa233b0136","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-20 17:51:12.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-03-729603-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"5f0ed2cb954b476fb3a473cfcc4f5c8f2fb3bec0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-20 18:12:33.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/d2a0d67 : FAILURE in 4s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/65d618e : FAILURE in 4s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"9c4125d17e4a556ef6e5f39947e277f9c16c49df","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-05-20 18:38:16.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/03/729603/2/check/Cinder-AS13000/8754e8b : FAILURE in 1h 17m 36s\n- cinder-test-fc http://218.29.111.78:9000/03/729603/2/check/cinder-test-fc/51422ba : FAILURE in 51m 26s\n- cinder-test-iscsi http://218.29.111.78:9000/03/729603/2/check/cinder-test-iscsi/6a9c248 : FAILURE in 1h 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b913e7423f16cf1bc09f9463b0ff46c9e3b22b84","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-05-20 19:09:57.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/03/729603/2/check/DellEMC_VxFlexOS/0bc449e/DellEMC_VxFlexOS/17 : SUCCESS in 2h 21m 32s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"c41fac90c95b865c6291c4b3d79f786284313846","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-20 19:15:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/03/729603/2/check/DellEMC_VNX_iSCSI/55eefb9/DellEMC_VNX_iSCSI/13 : SUCCESS in 2h 26m 54s\n- DellEMC_VNX_FC http://publiclogs.emc.com/03/729603/2/check/DellEMC_VNX_FC/5ae59c9/DellEMC_VNX_FC/13 : SUCCESS in 1h 06m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"09cc3c8e9c41fbfe7dfbf676d842447c991cca5a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-05-20 19:19:14.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/729603-2/Seagate-iSCSI : SUCCESS in 1:01:03\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/729603-2/Seagate-FC : SUCCESS in 55:12.25","accounts_in_message":[],"_revision_number":2},{"id":"2a2a7140cc833ea4d941c470d6469a45fb3caf49","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-20 19:20:42.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/2/check/DellEMC_Unity_FC/6d5fda4/DellEMC_Unity_FC/29 : SUCCESS in 1h 27m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/2/check/DellEMC_Unity_iSCSI/ebcc106/DellEMC_Unity_iSCSI/29 : SUCCESS in 1h 49m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"40f31bbf1518923ec93e526b6fa9e79574de9963","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-20 19:46:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/729603/2/check/DellEMC_XtremIO_FC/1b179d2/DellEMC_XtremIO_FC/2 : FAILURE in 5h 53m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/729603/2/check/DellEMC_XtremIO_iSCSI/5f3a62e/DellEMC_XtremIO_iSCSI/3 : SUCCESS in 1h 49m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"87a3aaac3691ffd77e8d19756c6d3264815b36d8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-05-20 20:33:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/03/729603/2/check/DellEMC_SC_FC/749a179/DellEMC_SC_FC/17 : SUCCESS in 1h 09m 56s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/03/729603/2/check/DellEMC_SC_iSCSI/7c92ee0/DellEMC_SC_iSCSI/16 : FAILURE in 2h 55m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"9fd2860db563eae7494835f9dcb3a8eafab79c76","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-20 23:51:45.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/2/check/tempest-dsvm-full-bionic/5586dc7/ : SUCCESS in 1h 23m 21s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/2/check/devstack-plugin-ceph-dsvm-bionic/bbf3416/ : FAILURE in 37m 45s","accounts_in_message":[],"_revision_number":2},{"id":"65316c0659e35c186ed9dc08992bdc6426459395","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-05-21 02:34:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nPlease add release note.","accounts_in_message":[],"_revision_number":2},{"id":"1f5bf36abca4daf5914c2ee58c3a1c135da1e55b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-05-21 03:23:50.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/192/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7e20b6e01a56c4bc63a6d2d6b2aa7483d13590ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2020-05-21 03:24:19.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* ibm-gpfs-dsvm-tempest-full https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/193/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e03887456241900f08ff2c2b74cebb1b740339f1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-21 06:37:40.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"17904eed1d2e793fbe5ab1994e7754377e81d1fc","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-21 06:37:47.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"b3fb6d625772c59abd7180938c55e1ff413ecdf5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-05-21 09:36:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/2/20200521155230/NEC-ISCSI : SUCCESS in 2h 28m 21s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/2/20200521155230/NEC-FC : SUCCESS in 2h 44m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5a690042f7b2d7c7e1594cb0c1876b50c5a87f6c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-21 10:32:28.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"48b227e01aa9b45c005ce41f243e0535feec9145","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-05-22 07:45:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1035/index.html : SUCCESS in 1h 11m 52s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1184/index.html : FAILURE in 1h 59m 55s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1217/index.html : SUCCESS in 1h 29m 17s","accounts_in_message":[],"_revision_number":2},{"id":"930ad5ebc776ad3f797064009d4b25caeb5479a2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-05-23 19:23:49.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/03/729603/2/check/DellEMC_PowerMAX_FC/e27d1e9/DellEMC_PowerMAX_FC/17 : SUCCESS in 6h 38m 05s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/03/729603/2/check/DellEMC_PowerMAX_iSCSI/f0780d6/DellEMC_PowerMAX_iSCSI/17 : SUCCESS in 8h 55m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"8093eac7b3198d74d361e9b5e45bfb7a4f15a776","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-05-24 20:23:29.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c18e4ac70e3a099ff96b01052581458d4eb824df","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-01 02:41:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c3e79318813f8fbb5320fa6a91cb1f5c59098a0a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-01 02:58:45.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/3/check/tempest-dsvm-full-bionic/9f7efa1/ : FAILURE in 11m 28s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/3/check/devstack-plugin-ceph-dsvm-bionic/17f3eab/ : FAILURE in 13m 06s","accounts_in_message":[],"_revision_number":3},{"id":"f773ba9bba45a81bd84becc63618eedb87236896","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 03:13:12.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/bf8c9aa965ebbcc6ba94a8f9ca18b031d6efaada/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"32ffac7344d8c7bd8f439bcf6a61a90556427cd3","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-06-01 03:23:54.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/03/729603/3/check/Cinder-AS13000/64c45cf : FAILURE in 3m 52s\n- cinder-test-fc http://218.29.111.78:9000/03/729603/3/check/cinder-test-fc/a723d52 : FAILURE in 37m 22s\n- cinder-test-iscsi http://218.29.111.78:9000/03/729603/3/check/cinder-test-iscsi/c443104 : FAILURE in 37m 31s","accounts_in_message":[],"_revision_number":3},{"id":"74d94d1f4d4d908c3b5a3ded522a1d1237026a6b","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-01 03:27:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"63635500ebf25b4f99677183fd657af6c7a5787f","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-06-01 03:30:56.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"68b3566778cd19ba03946af2dcb95dd55d28d810","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-01 03:32:40.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c46f27d6fc437d75e9bbdf695ee89057a0b88948","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-01 03:44:00.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/4/check/tempest-dsvm-full-bionic/823bbda/ : FAILURE in 11m 55s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/4/check/devstack-plugin-ceph-dsvm-bionic/ba82228/ : FAILURE in 12m 27s","accounts_in_message":[],"_revision_number":4},{"id":"cf22d43578cf9c06a8b06f58aff19848a2983fe6","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 03:55:25.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/ead6bf08203f37c309043e43f8edbe2736ebacc9/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e3166a18a4b524119a00fd0f81fd7c167ca2184b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-01 04:18:30.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"f35ead35a34688558bd6c2378ec3abe67d1861e1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-01 04:23:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/4/iscsi/ : SUCCESS in 55m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/4/smb/ : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":4},{"id":"a5fc204b1bcdb68793b5d88aae33dcc1a468b185","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-01 04:30:38.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2f6329078de40b87f3dd931cbda9da2b54efe8ce","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-01 04:33:32.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"f60fc32dac7284b57a495859ac347eba19e282d0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-01 04:34:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/4/check/dsvm-tempest-cisco-zonemanager-job/62d5a99/ : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":4},{"id":"6109433714f60c1f133a12cbe8890e35c4ae4b57","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-01 04:39:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/729603/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a93177c/a93177c.html : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":4},{"id":"27e551fdab26e202fc2c5eea84791eaefa8a1fed","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-01 04:48:32.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"852d03edaed1de2308ed47d237e2780930d3f70d","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-06-01 04:59:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/03/729603/4/check/nimble-iscsi-driver-dsvm-volume/369e184 : FAILURE in 1h 22m 43s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/03/729603/4/check/nimble-fc-driver-dsvm-volume/d7eead9 : UNSTABLE in 1h 31m 25s","accounts_in_message":[],"_revision_number":4},{"id":"6d8022a76fe4c67915445775a3cb894b7edbe46f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-01 05:01:54.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/4/check/DellEMC_Unity_FC/248145f/DellEMC_Unity_FC/82 : SUCCESS in 1h 10m 48s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/4/check/DellEMC_Unity_iSCSI/4d81c66/DellEMC_Unity_iSCSI/78 : SUCCESS in 1h 33m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"a3d01693b639ccfb9044fc139cfd840460502b34","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-01 05:03:49.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-729603-3/ns5_nfs : SUCCESS in 2h 6m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-729603-3/ns5_iscsi : SUCCESS in 2h 21m 35s","accounts_in_message":[],"_revision_number":3},{"id":"04a19526010275bd11055baa0364f7c2a931b2ad","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 05:15:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/3/20200601114149/NEC-FC : SUCCESS in 2h 24m 39s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/3/20200601114149/NEC-ISCSI : SUCCESS in 2h 33m 50s","accounts_in_message":[],"_revision_number":3},{"id":"43f605e8812fa1f9e72b15fae802ff41621e57dd","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-01 05:21:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/0f1ffe3 : FAILURE in 1h 51m 17s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f1eebfb : SUCCESS in 55m 55s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"a3e0935cd8270d8195178e991b127a610a6137c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 05:27:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a10c67e1695544749851e06d008d21f6 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde03ddce79b4c32b2b2053cb1dccb1c : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0281084be2074fc0b428aff6d980f653 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/537bf819ee9b400e8e9bcd4be828bffc : SUCCESS in 10m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc04ad6ea9af41bba858587298d6bff4 : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/c657c69afa954b74812d5d80031aaca9 : SUCCESS in 54m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0703c681992347aab40235d08ad873b3 : SUCCESS in 1h 09m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/29de205c45304d90beb58057c206a457 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b897f9231e474330a493cef3dd88cab6 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7e852691919439da7a620241d722eb4 : SUCCESS in 7m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d87d20aa97844d87bff7ff53f4106cd8 : SUCCESS in 50m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7985094a4f3a4affb487e1e578dbf502 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/305e993aae294eee8e1c700945f4f063 : SUCCESS in 1h 22m 52s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/655797105ada490eb9a413b68b96517d : SUCCESS in 1h 16m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b26f6a31085d469496afcb74bd7e7a30 : SUCCESS in 1h 35m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3de3d2c6c3674f51a86c4458d20c1b69 : SUCCESS in 1h 14m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3d2b18cdab3a4127b9f0f9492487046c : SUCCESS in 1h 12m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9fd987cfec6a499f9610440a4696919c : SUCCESS in 1h 58m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2978aaf57f44c0fa8a49bb837e34af8 : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":4},{"id":"9f564b5c4ce379f0681c24722ccd5821556cb506","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-01 05:40:15.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e (2 comments)\n \u003e \n \u003e Please add release note.","accounts_in_message":[],"_revision_number":4},{"id":"8100fd08003849f26f36a09164a3244d32fe78f0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-01 05:40:48.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-729603-4/ns5_nfs : SUCCESS in 2h 9m 49s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-729603-4/ns5_iscsi : SUCCESS in 2h 12m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c5daee6a328325ec56904279420678e957037cdc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-01 06:04:23.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/03/729603/4/check/dsvm-tempest-infinibox-fc/cb6c786 : FAILURE in 1h 22m 25s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/03/729603/4/check/dsvm-tempest-infinibox-iscsi/1093c6a : FAILURE in 1h 07m 50s","accounts_in_message":[],"_revision_number":4},{"id":"6e491dcf410e7c99007e877c20853049d5d55f0a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-01 06:06:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/729603/4/check/cinder-storpool-tempest/ab1732b/ : SUCCESS in 2h 34m 45s","accounts_in_message":[],"_revision_number":4},{"id":"35eda30d9c8218b93b5a9c13d3a3f155c808be8d","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-06-01 06:19:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/03/729603/4/check/synology-dsm-cinder-iscsi/6239719 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":4},{"id":"9686bf7a70cf652d203a932f12a98f354e9ced8e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-01 06:29:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/729603/4/upstream-solidfire-check/cinder-solidfire-iSCSI/79398ea/ : SUCCESS in 1h 00m 32s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"db62794c9f7c78b5da4d9a7af62115881e760577","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-01 06:34:42.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-03-729603-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"0c745f684d9f279dc372e523fbe147a8c73a17c0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-01 06:37:30.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e3c58b99baf70ebc954580b0ab0168075e80cad1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 07:44:43.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/03/729603/4/check/DellEMC_VxFlexOS/6ecd519/DellEMC_VxFlexOS/73 : SUCCESS in 2h 16m 25s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"47be2167a0fca57e8533aa1faf348473bb9ad68f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 07:58:04.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/03/729603/4/check/DellEMC_VNX_iSCSI/612c23d/DellEMC_VNX_iSCSI/64 : SUCCESS in 2h 29m 47s\n- DellEMC_VNX_FC http://publiclogs.emc.com/03/729603/4/check/DellEMC_VNX_FC/451d701/DellEMC_VNX_FC/65 : SUCCESS in 59m 55s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"942a339a575128af8504ebe73336b5086599de5a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 08:01:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/4/20200601141545/NEC-FC : SUCCESS in 2h 44m 57s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/4/20200601141545/NEC-ISCSI : SUCCESS in 2h 46m 12s","accounts_in_message":[],"_revision_number":4},{"id":"c4dd5a05d58085198cbe1428f4c94efed06dd7b3","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-06-01 08:17:52.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/03/729603/4/check/DellEMC_SC_FC/4558445/DellEMC_SC_FC/75 : FAILURE in 1h 12m 46s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/03/729603/4/check/DellEMC_SC_iSCSI/8454dfb/DellEMC_SC_iSCSI/62 : FAILURE in 2h 49m 36s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"fbd1c08933d2ea1d0684f13a4d6d11617686ea58","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-01 08:19:19.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/4/check/DellEMC_Unity_FC/698f45f/DellEMC_Unity_FC/84 : FAILURE in 1h 27m 46s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/4/check/DellEMC_Unity_iSCSI/cfab16b/DellEMC_Unity_iSCSI/80 : SUCCESS in 1h 37m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e1717e976a3c72fc5048549c906bbbb9d00e8b6c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-06-01 08:35:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/03/729603/4/check/fujitsu-eternus-dx-iscsi/5cc4b11 : ABORTED in 5h 00m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/03/729603/4/check/fujitsu-eternus-dx-fc/ad8b335 : ABORTED in 5h 00m 03s","accounts_in_message":[],"_revision_number":4},{"id":"43a910ec9cba6f204c78921a4abf7fe9ce77ef90","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 08:55:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/729603/4/check/DellEMC_XtremIO_FC/da9c63e/DellEMC_XtremIO_FC/93 : SUCCESS in 1h 28m 24s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/729603/4/check/DellEMC_XtremIO_iSCSI/09d7634/DellEMC_XtremIO_iSCSI/93 : FAILURE in 5h 27m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"9db4c1b9840e50ba88daa44735299829dd8db3f7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-01 11:08:16.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"1342393a4bf111b018cf9aee837d6caa62528d0a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-01 11:15:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/729603/4/upstream-check/cinder-cDOT-iSCSI/5cee7b2/ : SUCCESS in 46m 21s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/03/729603/4/upstream-check/cinder-cDOT-iSCSI-IPv6/c230ca7/ : SUCCESS in 1h 24m 10s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/729603/4/upstream-check/cinder-cDOT-NFS/52a2bce/ : SUCCESS in 1h 28m 48s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/03/729603/4/upstream-check/cinder-cDOT-NFS-IPv6/840a636/ : SUCCESS in 2h 00m 14s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/729603/4/upstream-check/cinder-cDOT-FCP/9e53501/ : FAILURE in 1h 19m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"018d9b027d6669879a9dbcea9191dc62ccf712fa","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-05 02:42:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3d85c7f9f9cbb9f648d74685f6001d8801d1d661","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-05 02:57:39.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/809d33a0c2395a4448f13863648c418aca7dc72d/index.html : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"78e4808afc463e261701d0b95f5dc355885bdc4d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-06-05 03:29:02.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/03/729603/5/check/fujitsu-eternus-dx-iscsi/ae698dc : FAILURE in 38m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/03/729603/5/check/fujitsu-eternus-dx-fc/7304491 : FAILURE in 28m 35s","accounts_in_message":[],"_revision_number":5},{"id":"2a63b0935db61233c026c5cbca5c9ca629de99fa","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-05 03:34:27.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"030fb5a5bde73409dfcfc91294deb6f423637850","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-05 03:39:59.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/729603/5/check/cinder-storpool-tempest/12cd32c/ : RETRY_LIMIT in 11m 31s","accounts_in_message":[],"_revision_number":5},{"id":"d59de38c5b45738b5281fb9dbc18a775299aa41e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-05 03:44:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/5/iscsi/ : SUCCESS in 1h 01m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/5/smb/ : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":5},{"id":"5bfd0fc6a35b38340a3e40ac87fe9cc8ab39e6af","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-05 03:51:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/729603/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/6da6aea/6da6aea.html : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":5},{"id":"960c401cbe2623923b55b65ac7e575cf8d62dfb7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-05 04:16:37.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/5/check/tempest-dsvm-full-bionic/d63d952/ : SUCCESS in 1h 25m 19s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/5/check/devstack-plugin-ceph-dsvm-bionic/2223d5b/ : SUCCESS in 1h 23m 45s","accounts_in_message":[],"_revision_number":5},{"id":"9220b45d7d325fed093e05d8cc9fd33d1a58e78e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 05:04:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7241d8bd7816421da69be18053f3bc4f : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b56f1927f240199e3fe472ca0aab2b : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d102dd0a1ba547e7841a3fcd23e626f9 : SUCCESS in 12m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f1d0a92feaa44c39687b50416a0b65a : SUCCESS in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8facb114ff084c7c926db0072898aca2 : SUCCESS in 16m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/63edb8dfa4e44386badbd4610c6069b4 : SUCCESS in 1h 24m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aada4b082ebe45d5b4b6c1d990e790e5 : SUCCESS in 1h 42m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/28dbe0d1608b4e57a507f8cf66b0eba7 : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc972c7650a3446f913ed55315554dbb : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d803811f933c46c3817822eb5cdbc479 : SUCCESS in 12m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6fe79170626d463187c2264f1390315f : SUCCESS in 1h 00m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f2c6e1cbf0641a1a3746c1a64693e9f : FAILURE in 10m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/94c94d30a7eb41049c9f3de8fce2d589 : SUCCESS in 1h 33m 32s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/95332b4897844413a147acfb512d6c0b : SUCCESS in 1h 19m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9326b33ca8614b258d383eaf39c7eead : SUCCESS in 1h 12m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ddb0f83d8736469f9231baef06e1c3f1 : SUCCESS in 1h 21m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ddaa12ad066648829193f5f4f90f1477 : SUCCESS in 1h 43m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b18a732ef744a0d919b7dbadc0ae670 : SUCCESS in 2h 19m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33537f17c24249688beb29c47d1461cd : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":5},{"id":"d3b363458a1e8ea0fa18600efe494494a92f5f86","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-05 05:06:02.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-729603-5/ns5_nfs : SUCCESS in 2h 22m 50s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-729603-5/ns5_iscsi : SUCCESS in 2h 8m 59s","accounts_in_message":[],"_revision_number":5},{"id":"0f98a9c387f8db7c0c5e7772c3e3b0113f90f6b4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-05 06:05:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/729603/5/upstream-solidfire-check/cinder-solidfire-iSCSI/43cdace/ : SUCCESS in 59m 32s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"737e0d80898ee472191c0b76ed901f19478d80a3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-05 06:34:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/729603/5/upstream-check/cinder-cDOT-iSCSI/0e88527/ : SUCCESS in 1h 14m 12s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/03/729603/5/upstream-check/cinder-cDOT-iSCSI-IPv6/5dc0a00/ : SUCCESS in 1h 28m 12s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/729603/5/upstream-check/cinder-cDOT-NFS/ec7afd8/ : SUCCESS in 44m 18s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/03/729603/5/upstream-check/cinder-cDOT-NFS-IPv6/e220a11/ : SUCCESS in 1h 24m 03s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/729603/5/upstream-check/cinder-cDOT-FCP/cc9d3db/ : SUCCESS in 1h 03m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"511b6f52a8da8b34f90e3391321d783446d3aabe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-05 08:29:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/5/check/dsvm-tempest-cisco-zonemanager-job/3c5e5c3/ : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":5},{"id":"34e48b58c90aeaa571e9a3ad2b49339cd3932d62","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-06-05 11:43:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/03/729603/5/check/nimble-iscsi-driver-dsvm-volume/a8cffa5 : SUCCESS in 1h 49m 52s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/03/729603/5/check/nimble-fc-driver-dsvm-volume/24af64e : SUCCESS in 1h 28m 43s","accounts_in_message":[],"_revision_number":5},{"id":"ac52b6330add6d0e95b392b1e758af4fe4c3a840","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-05 14:29:42.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/5/check/DellEMC_Unity_FC/684cd0a/DellEMC_Unity_FC/132 : SUCCESS in 1h 17m 04s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/5/check/DellEMC_Unity_iSCSI/acecb63/DellEMC_Unity_iSCSI/126 : SUCCESS in 1h 48m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"87d1ec2247cac0cfff10c6efba7c7e1d08bf28c2","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-05 19:09:07.000000000","message":"Patch Set 5: Code-Review+1\n\nYou should add release notes about this change as Ryan said before.\nLooks good otherwise.","accounts_in_message":[],"_revision_number":5},{"id":"776b2adcb63fa742f7ed86614c82900e9e20882f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-05 23:47:03.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/5/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c36da11 : SUCCESS in 58m 23s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/5/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f42deca : FAILURE in 1h 53m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"4e0b6741a9441e93a9766043f6163426898a4329","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-06 01:23:31.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/03/729603/5/check/DellEMC_VxFlexOS/5af5cbb/DellEMC_VxFlexOS/98 : ABORTED in 4h 00m 31s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"281248f8b3505b0e1c1d102996ef636a1723faf8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-06-06 04:07:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/5/20200606103201/NEC-ISCSI : SUCCESS in 2h 19m 19s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/5/20200606103201/NEC-FC : SUCCESS in 2h 35m 48s","accounts_in_message":[],"_revision_number":5},{"id":"d5b365002ed6d9f678b00d0c319a3a493f5dba44","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-06 10:02:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1156/index.html : SUCCESS in 1h 11m 57s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1307/index.html : SUCCESS in 1h 39m 35s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1334/index.html : SUCCESS in 1h 23m 27s","accounts_in_message":[],"_revision_number":5},{"id":"6349c2ebbfa95cda04035dea5fe394fccc184053","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-06 22:00:37.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"787e1b5a958ef45d35c7e7c6681692e41ef491f6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-06 23:54:04.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"2c9d412808fcdaf007022d8ba3bbfb69dc38e0c7","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2020-06-09 12:45:43.000000000","message":"Patch Set 5:\n\nrun-storpoolci","accounts_in_message":[],"_revision_number":5},{"id":"16dadffa0be3a4a26b637734c68b81c953e0bee7","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-09 15:01:49.000000000","message":"Patch Set 5: Code-Review+1\n\nPlease add release notes","accounts_in_message":[],"_revision_number":5},{"id":"0cf3d16c3f0720de96eb652c62504f4e83e818ee","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-12 05:39:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6ccb43a44c3f20cd38670c48aeb9e42ea8de53bb","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-12 05:53:38.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/3ff06ddd125e20369c6212e49b3a47d9de453aac/index.html : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"3510a76c8bf0e98673b5fca56cccb8b873072a5d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-12 06:11:10.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/6/iscsi/ : FAILURE in 28m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/6/smb/ : FAILURE in 30m 58s","accounts_in_message":[],"_revision_number":6},{"id":"33d7cc3f70cb0855b0d85fc937b5d8b782706c21","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-12 06:17:08.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/03/729603/6/check-not-vote/ext-cinder-zuul/863dc59 : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":6},{"id":"13f8e44bf5f0a4e666c81b4eb5b7eab65c1f9eb0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-12 06:22:18.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-729603-6/ns5_nfs : FAILURE in 42m 10s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-729603-6/ns5_iscsi : FAILURE in 34m 30s","accounts_in_message":[],"_revision_number":6},{"id":"93ebb4ba4506407767134a4b3e5d131a97b773da","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-12 06:31:13.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"8cd4774343354cb37b4da56854a4c57e99ff38a6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-12 06:46:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/6/check/dsvm-tempest-cisco-zonemanager-job/4d44f9b/ : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":6},{"id":"567682628c9cea7d822d9a2cc0291aaaf7070ec5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-12 06:48:07.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/729603/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/474143c/474143c.html : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":6},{"id":"66765787ee50f8ccf43322bebe8645cf4b3c44e4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-12 07:12:01.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/6/check/DellEMC_Unity_FC/3dd3417/DellEMC_Unity_FC/168 : SUCCESS in 1h 32m 06s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/6/check/DellEMC_Unity_iSCSI/caa421a/DellEMC_Unity_iSCSI/161 : SUCCESS in 1h 31m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"83eb30a8e2f1b384f33a335d19c40d4ece170082","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-12 07:16:19.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/729603/6/check/DellEMC_XtremIO_FC/2371634/DellEMC_XtremIO_FC/190 : SUCCESS in 1h 32m 27s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/729603/6/check/DellEMC_XtremIO_iSCSI/4a044a9/DellEMC_XtremIO_iSCSI/187 : SUCCESS in 1h 36m 19s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"d51375d1dd8927756959f83d3800ed8747ada46c","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-06-12 07:22:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/03/729603/6/check/nimble-iscsi-driver-dsvm-volume/8bc8b64 : SUCCESS in 1h 42m 04s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/03/729603/6/check/nimble-fc-driver-dsvm-volume/1d38b3b : SUCCESS in 1h 32m 56s","accounts_in_message":[],"_revision_number":6},{"id":"19a4f2969c81e2a371eb23634ed77a12ee6f569f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-12 07:31:54.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/tempest-dsvm-full-bionic/997a988/ : SUCCESS in 1h 26m 07s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/devstack-plugin-ceph-dsvm-bionic/eb952a1/ : POST_FAILURE in 1h 47m 31s","accounts_in_message":[],"_revision_number":6},{"id":"9b334ae49679b7b1424501d3aca741f6fbff5e2e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-12 07:33:11.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/2dbcefe : SUCCESS in 1h 05m 17s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/769fe18 : FAILURE in 1h 53m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"333929af8804e584492d0ec5e98cb58e5b906be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 07:54:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/559915a86f7e4c5cba356e85e107ebc9 : FAILURE in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0670677d7b5449179570ae56e16a18b1 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a796c58943764e0fab8af8466cb925c9 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5052cc1a9c649468cb50bf961eb5244 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529b9e1926b04dfc9370ec6111b5ade3 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/039b6b5cdd3542d5a13a77c7b7add44c : SUCCESS in 1h 00m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7fb2f58d8dbb44c38490c5aca73efd0f : SUCCESS in 1h 23m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41d57feffb9247fbb9854d87dfb99fe8 : SUCCESS in 6m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab2a1a39b6444f83925229180f2cfa50 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/031a1a66d0ce464e9f225388d199e5b9 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5f0d1d8adca45fd825d0d1a00b1d16a : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a2db211adbb14181b6741fc6b464f0c9 : FAILURE in 32m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6db267562b8a473687e5f8d878ef9078 : FAILURE in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a9af99bc03d04cf0b1cb74b2b70b0dc3 : SUCCESS in 1h 08m 51s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d99c1764c2f54e1fa6f397f5239ee669 : SUCCESS in 1h 39m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcc730b238a34ccf9fd43ea16588bb04 : SUCCESS in 1h 24m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7fb62c56de034b008ee7b77f105170fb : SUCCESS in 57m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1583dc3a8f184a7fa71092e2ca569082 : SUCCESS in 1h 22m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fcc1e85d2ac430ca3a88aa671d02da7 : SUCCESS in 2h 08m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2737dfad4bf42d9af4db8faeeb01fba : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":6},{"id":"a8f69dbb706e9835ffe8c68d1e63536aec1a0353","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-06-12 08:20:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/6/20200612143944/NEC-ISCSI : SUCCESS in 2h 35m 26s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/6/20200612143944/NEC-FC : SUCCESS in 2h 40m 33s","accounts_in_message":[],"_revision_number":6},{"id":"be7d283dec6fdc6c8777624ff8ba86bed6c5d480","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-12 08:50:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1183/index.html : SUCCESS in 1h 14m 00s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1333/index.html : SUCCESS in 1h 37m 10s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1358/index.html : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":6},{"id":"79836c951c35a15b669331666374a371074556b7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-06-12 09:34:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/03/729603/6/check/fujitsu-eternus-dx-iscsi/17c3c41 : FAILURE in 3h 49m 44s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/03/729603/6/check/fujitsu-eternus-dx-fc/dd5e829 : FAILURE in 3h 33m 33s","accounts_in_message":[],"_revision_number":6},{"id":"1e7acb11acf6788f080e30fc509fdca58c9a4db3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-13 02:15:22.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"1be28531c56dd0c77184049e7b964f2eba047552","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-13 03:19:35.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"1f3107b773a78ee6663d694f9457df0ff1782377","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-14 20:09:58.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/03/729603/6/check/DellEMC_PowerMAX_FC/75465e0/DellEMC_PowerMAX_FC/101 : FAILURE in 4h 10m 39s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/03/729603/6/check/DellEMC_PowerMAX_iSCSI/f0e91ea/DellEMC_PowerMAX_iSCSI/85 : SUCCESS in 5h 50m 15s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"46da59d0b4430847fa84200b16f2028823c5c2a6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-14 20:35:09.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"e39a86c1c6f48462c4789358918c2b10acea5871","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-15 01:01:19.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3b279a4b3fdd73443a326b38ee6cb2ece23a1560","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-15 01:02:11.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"175915f169f5bf8d66c6a9613bffd8d2ac52868a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-15 01:16:32.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/72072d6 : FAILURE in 14m 52s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/aaa8297 : FAILURE in 14m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"1113eaf336c6a0345073305e36dbe984eb98a7cb","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-15 01:20:48.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1196/index.html : FAILURE in 16m 29s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1345/index.html : FAILURE in 19m 07s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1371/index.html : FAILURE in 16m 47s","accounts_in_message":[],"_revision_number":6},{"id":"5b086c1fed75128844cb24ee03ec416cee47338d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 01:30:37.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72706cbbe968416a8a884cbc0e5d1bcf : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c90151ffe4d94831940a65a1d6fe620a : SUCCESS in 10m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1bff5ed16b74d84ab67a539836c8b58 : SUCCESS in 11m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d5ec2f38915418293b61cbd82ebcacb : SUCCESS in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1fe88602d914b62b91e2afb4575171c : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/e9fbc78a45cd4a39ad7ab012a64aa166 : FAILURE in 16m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3be1460df5a44a9fb72234a3139bce62 : FAILURE in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79dab72fe1f24238b4a27f9da38fd80d : SUCCESS in 8m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d2ae991b6524c40a7f8869d3fc28b8f : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a85838f4d59443929a4b45d9ce37ef4f : SUCCESS in 6m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f64a12c260f54394a76ce6af835247d8 : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fab00d090e604ca79efdbe6f0a8287ed : FAILURE in 10m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f00f3ebbea9a4e9285007dc5f4b4e6d1 : FAILURE in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f19ea2ea88214736b9ddf24aa4db7f6a : FAILURE in 12m 34s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/edaf3ca6e4fc4a18a196618c486555af : FAILURE in 10m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eafc9a77e82b46788a225ecacf04a338 : FAILURE in 19m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5fb97e76a3d64f9abd1a89bf62a58b8b : FAILURE in 10m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/24c49333fe964b46b19b00e4abe345ed : FAILURE in 10m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/757bdeded4b74efdb345b070adf64948 : FAILURE in 12m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f592adde76b414c8335e80c4dba2e40 : FAILURE in 7m 04s","accounts_in_message":[],"_revision_number":6},{"id":"d8dde264c3f074755eaeaf9b1eea54efe5831a52","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-15 01:33:18.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/tempest-dsvm-full-bionic/e8e357c/ : POST_FAILURE in 25m 11s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/devstack-plugin-ceph-dsvm-bionic/c7d2472/ : POST_FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":6},{"id":"65300043415ace012b3b182838c72a996cb21153","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-15 01:33:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/6/check/dsvm-tempest-cisco-zonemanager-job/e9f124d/ : FAILURE in 24m 49s","accounts_in_message":[],"_revision_number":6},{"id":"836a9262212002c36c246c4e6c82fe0d556fdc52","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-15 01:37:49.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/6/iscsi/ : FAILURE in 36m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/6/smb/ : FAILURE in 35m 22s","accounts_in_message":[],"_revision_number":6},{"id":"175d3732348331e488c91f5c02fe9c81512aad9e","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-15 01:44:18.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"bd71f7b9fd84a38fdac83877980963d755d18784","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-15 01:53:00.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"30ccdf4d64dd1c3c016759349cdddb93526e7581","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-15 01:58:48.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"050d173cf432e5c2956dae3a7798a5e1fc38b31a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-15 01:59:35.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f501136 : FAILURE in 14m 36s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/65d0667 : FAILURE in 14m 58s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"9a980303abd4d9a3d0caa164e9776b690a3e1779","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-15 02:04:14.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1197/index.html : FAILURE in 15m 47s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1346/index.html : FAILURE in 19m 31s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1372/index.html : FAILURE in 16m 37s","accounts_in_message":[],"_revision_number":6},{"id":"b92a9efecdb0de199a96984579a283bcaf582e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 02:05:27.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18589e3dfcde4fe08da17b9a7818c7bb : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/447d275a7af74469b729821bba267cbf : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a20888e0704248a4b6e7e597d6e28687 : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e4654dce26e4352b0ccbd9609743062 : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0b4ddad0f142b1a3962197f3fc5919 : SUCCESS in 15m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d32c65693d3f4403876cf90e7775297d : FAILURE in 11m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85905ba632554e79a8eb0305e0a8dbc4 : FAILURE in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52d3a614a4f24ce6abad52ee12199ed4 : SUCCESS in 7m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/abf788ea6f4541ab9cc7ffb299702f34 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/12fea8b6e8ea4063a6a3af449af8140b : SUCCESS in 7m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3912d745054416c8c34ba3a071dcb5a : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/74bf2a21b8c5402e9f6f3909b80e2df4 : FAILURE in 9m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1dcfb5f3946e4b999192d73392cafceb : FAILURE in 8m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d861a789b00f41bba391bd0774bb6cc7 : FAILURE in 9m 55s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bd12c6b57d5c4310b87fc7f94befd381 : FAILURE in 9m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c82ed8154964b63abba806accc2d253 : FAILURE in 14m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/537d911ccafe4747b7ad70a4cb7ef58a : FAILURE in 10m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dbb32426d5284e1291f0fe8dda0548b8 : FAILURE in 9m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9872f3d426034f3a8051ad37ccd6f259 : FAILURE in 15m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/266e154d7ebd402ea4efd49da2e84a15 : FAILURE in 7m 53s","accounts_in_message":[],"_revision_number":6},{"id":"1601d7f6b521a140693ae818b2887efcd9bb4c6f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-15 02:14:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/6/check/dsvm-tempest-cisco-zonemanager-job/63261b4/ : FAILURE in 24m 51s","accounts_in_message":[],"_revision_number":6},{"id":"855cb6a05dfcb4c334ee9fc5ea82c0d4832421bb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-15 02:16:03.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/tempest-dsvm-full-bionic/c561df9/ : POST_FAILURE in 27m 47s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/devstack-plugin-ceph-dsvm-bionic/48be6a6/ : POST_FAILURE in 27m 53s","accounts_in_message":[],"_revision_number":6},{"id":"69d1c6e4487dd75c0a510e4487de2df8cec8b1b9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-15 02:23:29.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/6/iscsi/ : FAILURE in 35m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/6/smb/ : FAILURE in 38m 46s","accounts_in_message":[],"_revision_number":6},{"id":"ca6a98c1efc5317c3d327143258f9150e1673a4d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-15 02:41:10.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"115f9feb1fe71bd877bfa9891e48c1d971a1f741","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-15 05:14:38.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8768bdaf1a4ee2f34242df506f5e8e4d1b36c3f0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-15 05:29:48.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9934c64 : FAILURE in 14m 52s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/21ffd5d : FAILURE in 14m 49s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"8e4415c560b5c7a86dfab75addb22a3a9f39a106","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-15 05:33:47.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1199/index.html : FAILURE in 15m 57s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1348/index.html : FAILURE in 18m 51s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1374/index.html : FAILURE in 16m 50s","accounts_in_message":[],"_revision_number":6},{"id":"8af6b7257dfcf28682c6d7a953bce3960ccb8f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 05:36:26.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfa7176544484aa0b87476a1135343c4 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/842498a77573420185b8682cee98ca62 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c8d2d879f15484cafa87ed9a0996754 : SUCCESS in 14m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0b7a400598641dfa621fc33d6a432c2 : FAILURE in 15m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b7d609e5a644a0b85f24367b266894 : SUCCESS in 20m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/c519f6d3119e484c9babea0a28c5f2e0 : FAILURE in 18m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc1cd029a9a74c28bf084d80dba84adc : FAILURE in 16m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67195b2ee0494054ae19dedf251896e8 : SUCCESS in 9m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/acbae83c3faa4fa88c003b21cd4391f3 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86e99f370e24471386d82196a65443ac : SUCCESS in 11m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/903f93ce10c042ce8d42775c234334a5 : SUCCESS in 11m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c9ae9bf1861246b9811d02dd19461a23 : FAILURE in 18m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/583615ad473d44099082ff83af55d5eb : FAILURE in 9m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/353d5006695b498399e62cc3918f524b : FAILURE in 10m 51s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9b531b8ad5d74ddd8e53759b8bd60a47 : FAILURE in 18m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d2e225e26eb64c9bb350521a83318b17 : FAILURE in 16m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29e321a19eeb48aba6172b60fa5c2fc6 : FAILURE in 17m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a445dbdfa20a4c02962817f172b1e533 : FAILURE in 17m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1899f407cb8147bcb713123cca6e4afa : FAILURE in 12m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e8169798d574d25b07cda6f80a0df87 : FAILURE in 18m 05s","accounts_in_message":[],"_revision_number":6},{"id":"273e021b3f8c74db5b91619b53d898fb7e55166d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-15 05:39:55.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"13504535ff66d98f4d59dd7dc68f6c31a48c9abc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-15 05:43:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/6/check/dsvm-tempest-cisco-zonemanager-job/9c06bf9/ : FAILURE in 24m 18s","accounts_in_message":[],"_revision_number":6},{"id":"03ee1de635ddb31dabeacc2e0d2a0e36d6a6696d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-15 05:46:00.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/tempest-dsvm-full-bionic/1a23081/ : POST_FAILURE in 25m 02s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/devstack-plugin-ceph-dsvm-bionic/211d2b8/ : POST_FAILURE in 27m 18s","accounts_in_message":[],"_revision_number":6},{"id":"5422df05101e0498b48adb1693c878b44079d463","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-15 05:50:43.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/6/iscsi/ : FAILURE in 35m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/6/smb/ : FAILURE in 35m 39s","accounts_in_message":[],"_revision_number":6},{"id":"2311cda106b3d85d1deb74a6dd0f69f171681d7e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-15 06:05:54.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"19dcfbdd3f9e34b439ea583ee5fefdfefa366d72","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-15 06:10:02.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"b4eee91d32b6b86a4f4de2859e25cff3f3bd4830","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-15 15:05:19.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6eb8189c423547b753c7f1e0e52ded0e7ea85861","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-15 15:26:02.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/d4bd658 : FAILURE in 20m 25s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/0198c94 : FAILURE in 20m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"cab3f3581da08bece2da5125972166d5ae8e687c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-15 15:39:42.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"c401d2c7f3acb0bce478ffba6a1c350a35b1294c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-15 15:44:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/6/check/dsvm-tempest-cisco-zonemanager-job/418edad/ : FAILURE in 34m 25s","accounts_in_message":[],"_revision_number":6},{"id":"4f04140ce966d5944faaa335818fb863b817ae65","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-15 15:56:29.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"3301735d2cb01d8ea6b0949362183184edee5cc9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-15 16:23:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/6/iscsi/ : SUCCESS in 1h 17m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/6/smb/ : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":6},{"id":"edb0a7080a7fa55b748099d32a862516ca495a8c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-15 17:13:51.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/tempest-dsvm-full-bionic/9933f80/ : SUCCESS in 1h 45m 42s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/devstack-plugin-ceph-dsvm-bionic/8201343/ : SUCCESS in 1h 39m 57s","accounts_in_message":[],"_revision_number":6},{"id":"0c321b98b6c298f69fea936c5aacdc963f08ff82","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-15 17:31:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1205/index.html : SUCCESS in 1h 13m 50s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1354/index.html : SUCCESS in 1h 39m 44s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1380/index.html : SUCCESS in 1h 25m 14s","accounts_in_message":[],"_revision_number":6},{"id":"99830ef32e5ab6032d9f9fd51679e249f960d69e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-15 17:46:44.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"fdcecbefb6087c4666719b2cf625f399ed04384d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 18:39:52.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/245ab52e6a9f49b3a76a7ffb4fc3697f : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf21208dc494bfc8f41ff7d2b723e7f : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e29db574c154450b970071147d6f6ec : FAILURE in 16m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/811d31b14d754d339d155036182a76fb : SUCCESS in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0a9da5ac5645bf877f565ada7e2073 : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/681b06586cd940c091db872185761127 : FAILURE in 13m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8505083035547ab87115806e0b2aa9f : SUCCESS in 1h 14m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb5969cfe7954b0f88c34e19644a69ec : SUCCESS in 9m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/65e1f3cc365c410c8e2cb082368c6f4d : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9892abf857f4a858819bf7fffe91342 : SUCCESS in 9m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0d22936dd4543139134121c3181f7f6 : SUCCESS in 8m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c26c1f60f28d4e8d9c71acf58456a2f9 : SUCCESS in 32m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/327192656c12433896b929474bcbc797 : FAILURE in 7m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e1e915efa10e46788a3599168acb9d06 : SUCCESS in 1h 06m 45s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d28fb28bbab4c7fb49823c1e07bb879 : SUCCESS in 1h 29m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/217a1c29048847cd9134106dc8ae1375 : FAILURE in 24m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e2904171ca6743a29e6765334e4a1f7e : SUCCESS in 1h 14m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5dc040540df945a0b782f86f8935521f : SUCCESS in 1h 21m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/422a52df78a24c648274155efe6819a0 : SUCCESS in 2h 07m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a2c556da1a04c6591ace49fef07da92 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":6},{"id":"983a2c46c1682a590fe782f5c278b2936af9804d","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-16 01:44:29.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3174f21c9e1d3020127f20ef0038ff3df3075b02","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-16 02:10:54.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4033f59 : FAILURE in 26m 06s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/99bfc90 : FAILURE in 19m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"ca03a34653622ba0de08075449d99ea2d6b3d607","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-16 02:17:22.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"6c128d305c2f2eec7d0c621849cec2a65001cc33","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-16 02:26:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/6/check/dsvm-tempest-cisco-zonemanager-job/402671a/ : FAILURE in 34m 08s","accounts_in_message":[],"_revision_number":6},{"id":"9e82471afc043f9b5e0b681c969ceb264bfbd140","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-16 02:35:33.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"72463099f81fac920c12fcf7f56428b63d2cbae3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-16 02:52:20.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/6/iscsi/ : FAILURE in 1h 07m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/6/smb/ : FAILURE in 1h 06m 31s","accounts_in_message":[],"_revision_number":6},{"id":"974466cf3c23166de4b94a59d6978d910f8bb64c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-16 03:24:16.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"9aa90cd5b6b299160e75733137dc09c77aac66e5","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-16 03:34:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1209/index.html : SUCCESS in 1h 14m 29s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1358/index.html : SUCCESS in 1h 49m 59s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1384/index.html : SUCCESS in 1h 29m 02s","accounts_in_message":[],"_revision_number":6},{"id":"177256e6046a0198d006169ed9b38e348e9bc2dc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-16 03:35:10.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/tempest-dsvm-full-bionic/d210090/ : SUCCESS in 1h 45m 42s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/devstack-plugin-ceph-dsvm-bionic/3ea3ae5/ : SUCCESS in 1h 34m 45s","accounts_in_message":[],"_revision_number":6},{"id":"cf450e49d0ae7b86cbea2be72a1c4725a635a70e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 03:47:59.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4f7a10052d14eefbd3789f845c98d71 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2465e86d4d14eb388cefcb2d2e22bde : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b552d1692f449e5b887736956067183 : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62a6b79b9baf46eab48505e57605ce8e : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee2a693226f4c8e9809e348b62c5b6b : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/3016361b2a404797b20a87738566ceb3 : FAILURE in 14m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9899e673c3db454bb2d1b9757258db0a : SUCCESS in 1h 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4406f530af334d0b89aedb462d0c4000 : SUCCESS in 7m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/175da94f2fd64a8ebb39f0e60869371a : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97d1eb79c358471eb9baa199896527fd : SUCCESS in 5m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6945c23dcc95428990a43bfeb95569f3 : SUCCESS in 7m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/647f0a9a81b340f9b79564f70bea36c6 : SUCCESS in 39m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2b4901f010d424ba4f3a8cfc4ee64f9 : FAILURE in 7m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/95f21f8b46f445e3a6f801dc06b736a6 : SUCCESS in 59m 05s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ecdc29f90d8247d08f531b920ec1b6ff : SUCCESS in 1h 28m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/92297635a4b042cdaaf0f3f06b374ba9 : FAILURE in 21m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aeb44fa4df2c428b8deb7fad492d8c44 : SUCCESS in 1h 17m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25c4111cbbe248edb9ffe0828340cfa4 : SUCCESS in 1h 21m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d8185955f004c17b8c492712759362b : FAILURE in 2h 01m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8975e78615741518b0b1f8f5793013b : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":6},{"id":"f5f4340f936b269ebe51430442b2efa208f3c6f4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-16 04:08:38.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"2062f8ec1944375f2906c57617bdfab748a13d92","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-16 06:48:13.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/03/729603/6/check-not-vote/ext-cinder-zuul/3fd6f92 : FAILURE in 1m 00s","accounts_in_message":[],"_revision_number":6},{"id":"ec76b67d65b86bb9829a1df1b4f58bbae8bca3ed","author":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"date":"2020-06-16 09:26:10.000000000","message":"Patch Set 6:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":6},{"id":"e85b3eaa77446698d2de495028dc06343b26c9eb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-16 10:09:28.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"efc0938cfcdd63cc10038eb3691d59afd7afc094","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-16 22:40:07.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-729603-6/ns5_nfs : SUCCESS in 2h 21m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-729603-6/ns5_iscsi : SUCCESS in 4h 49m 8s","accounts_in_message":[],"_revision_number":6},{"id":"4fe024a9c6d2efa9241204049641c275e7064a3d","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-18 02:13:39.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"7e9409aa44dd80f781427f7b51073513057dd13a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-18 02:52:38.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/6/check/dsvm-tempest-cisco-zonemanager-job/df5cb63/ : FAILURE in 33m 47s","accounts_in_message":[],"_revision_number":6},{"id":"014cee3b17f94eda69a35a899290eeaaa679fcda","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-18 03:05:05.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"95a00eff5c10aa170a8770be7a45d9f46dd4ee13","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-18 03:10:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/6/iscsi/ : SUCCESS in 56m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/6/smb/ : SUCCESS in 52m 37s","accounts_in_message":[],"_revision_number":6},{"id":"a5f7031d306a8afe1dad22a65e43e178a69e8fb8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-18 03:17:23.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"b92fc77a0812151b1bfa6286c3febcab92c14f99","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-18 04:04:29.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/03/729603/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ea79896 : FAILURE in 20m 09s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/03/729603/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/2fa117d : FAILURE in 20m 03s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"c743002e69ed76009fce66bf13baf9b5bd15eeea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 04:20:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/859248254a0145d6a374a21711275d6d : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6e894b0eb341f2a16b2aa918a13305 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23eac8c469a74357a2b4821cd7f18e90 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/650ae4c84e004724826af6559ae76d0a : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb2c52afc424821abb595dd94d00319 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/a022b1619496474d9481e296a5751dba : SUCCESS in 1h 03m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3bbb5fe0b1447ab9aff90d03c52f5f3 : SUCCESS in 1h 08m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3581cc7fda544bf3abd87abc36e4a5b0 : SUCCESS in 7m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f47bac1cdd974b0bb0fb08d214d7f2bb : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9944e42182b54291bb2bb0c0ff785069 : SUCCESS in 7m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8ad6331cade42e194f7bf657baabab1 : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fa27364fb8aa4334a2d0b1eadd6ab2d4 : SUCCESS in 48m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/990c16a31aac4955a975dc82e728f87c : FAILURE in 5m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/10b3d40019c246ca8d26bf700c60eac5 : SUCCESS in 1h 08m 30s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d8c2a885346e4192af7ec5c9029f4d72 : SUCCESS in 1h 18m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a35856d5758c4b21b241ca4b208e371c : SUCCESS in 1h 11m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eb2b75718f8f4a218edadf52c1c2b8cf : FAILURE in 1h 27m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3523fb1d41044d4ebdcf85ee7e1de675 : SUCCESS in 1h 15m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b0bc81a41c9475f815af5d5913c5b29 : SUCCESS in 2h 04m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dafd160207d474eb42f519d205f8fef : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":6},{"id":"dea7228e79736d03eaf5abacf262c3317ab249e3","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-18 04:38:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/729603/6/upstream-solidfire-check/cinder-solidfire-iSCSI/3a118c0/ : FAILURE in 16m 42s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"54ace3df94227142c4d23f242c4ac42e29b6f8b8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-18 04:51:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/729603/6/upstream-check/cinder-cDOT-iSCSI/ff1892b/ : FAILURE in 16m 59s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/03/729603/6/upstream-check/cinder-cDOT-iSCSI-IPv6/86caad7/ : FAILURE in 29m 40s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/729603/6/upstream-check/cinder-cDOT-NFS/0d962a9/ : FAILURE in 16m 29s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/03/729603/6/upstream-check/cinder-cDOT-NFS-IPv6/36923a2/ : FAILURE in 29m 56s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/729603/6/upstream-check/cinder-cDOT-FCP/c221eda/ : FAILURE in 19m 51s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"6e5f35f2745eac79ef2182d9c0f932d7c72de1e2","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-06-18 05:40:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/03/729603/6/check/synology-dsm-cinder-iscsi/e69512a : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":6},{"id":"99da0236e0b01af0805e3f75295c76186fdba0b0","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-06-18 05:48:27.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/03/729603/6/check/DellEMC_SC_FC/56aab9a/DellEMC_SC_FC/121 : FAILURE in 21m 14s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/03/729603/6/check/DellEMC_SC_iSCSI/e90f893/DellEMC_SC_iSCSI/108 : SUCCESS in 1h 27m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"bf23fc5d547e98ce418e008c7105a36f27000ff7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-18 06:01:42.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/tempest-dsvm-full-bionic/5f3551b/ : SUCCESS in 1h 26m 28s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/6/check/devstack-plugin-ceph-dsvm-bionic/92de433/ : SUCCESS in 1h 33m 01s","accounts_in_message":[],"_revision_number":6},{"id":"d388867b6929f67a928f482225e021eedcb30f1e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-18 06:21:18.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"eaa6efcfe7d86d4dcbe45d3e1db1eb0d07bbfb5d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-18 08:43:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/729603/6/check/DellEMC_XtremIO_FC/c1ec8d2/DellEMC_XtremIO_FC/229 : FAILURE in 1h 34m 57s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/729603/6/check/DellEMC_XtremIO_iSCSI/f41e2e1/DellEMC_XtremIO_iSCSI/226 : SUCCESS in 1h 39m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"db9edb62ba3c0517521b3aaa61dd1b37a012f827","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-18 09:37:02.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"5dc39ec6a4ccea9934feacd92b4a7119c4dd87ca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-18 10:26:57.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/03/729603/6/check/DellEMC_VxFlexOS/fb46a06/DellEMC_VxFlexOS/150 : SUCCESS in 2h 07m 42s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"5e1d6b11545b65d460b9c61e5c08b2771051b5e1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-18 11:24:40.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"0a2c511d82626620a1fff2571145e1ad533db3d5","author":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"date":"2020-06-19 04:52:33.000000000","message":"Patch Set 6:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":6},{"id":"a408c38394767fb1588116069adc2b21a2e3aa56","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-19 05:31:11.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/03/729603/6/check-not-vote/ext-cinder-zuul/c62025a : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":6},{"id":"09f3f384efe44b1dc2f3a1ab44fecda962ab0692","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 05:35:37.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"62dbfc8031735f529d97f5ef2f0279b9b8c39390","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 18:32:08.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f311f7cb154464b310eb9899f94eff35e6a7ee3f","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-08-04 13:50:33.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9b4ffccf66976559136275eadedc1c2e221c84b5","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-09-23 11:27:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nHi Yong,\n\nThe code itself and the test coverage LGTM.\n\nPlease see a comment in the reno.","accounts_in_message":[],"_revision_number":6},{"id":"34b94ed316950b56dfe7d9e45f3a886587065af6","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-09-24 06:30:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4d86202ecd11c454f44d9c0683dbf866385c1d90","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-24 06:30:44.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/03/729603/7/check/fujitsu-eternus-dx-iscsi/be8c5af : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/03/729603/7/check/fujitsu-eternus-dx-fc/c0cdfa1 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"284a96cf3b488bb99ba92fbe8f1c2d9e5be44f68","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-09-24 06:37:05.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a1045b2b4dd0f3ea0c664dcc97ec34baa1a4a997","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-24 06:49:16.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/105c0855fa12da0ad2878261a136b1c7b0c830c9/index.html : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"624c1c9f4942a9245f64fec4f84e4a89dfa4c7e8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-24 06:51:39.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/03/729603/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/fb28e76 : FAILURE in 20m 56s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/03/729603/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/dbb21b7 : FAILURE in 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"d0245b25f4f97a93aedb930ab2623566abfc0056","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-24 07:05:15.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_FC/d114ad0/DellEMC_Unity_FC/732 : FAILURE in 34m 44s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_iSCSI/3097807/DellEMC_Unity_iSCSI/699 : FAILURE in 29m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"510f23c8874db6006e82fd04034e16fab5f321d5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-24 07:07:03.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_XtremIO_FC/5d9f94a/DellEMC_XtremIO_FC/993 : FAILURE in 36m 25s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_XtremIO_iSCSI/4bc0914/DellEMC_XtremIO_iSCSI/955 : FAILURE in 31m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"70ccb11c75e10ceee6cbe1da821a2bbec7d54b02","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-24 07:14:34.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/03/729603/7/check-not-vote/ext-cinder-zuul/95c18f8 : FAILURE in 43m 36s","accounts_in_message":[],"_revision_number":7},{"id":"6ca26ec3c3e45197dc450965e6b73f049d70f213","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-24 07:45:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/729603/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/91907a2/91907a2.html : SUCCESS in 1h 12m 31s","accounts_in_message":[],"_revision_number":7},{"id":"5c6bff248c35ccc028447d05f00a52db159f6650","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-24 07:51:59.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/729603/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"5e152206a8fefc93edc1f4e84cea16f205c76287","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-24 08:15:42.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/7/iscsi/ : FAILURE in 1h 44m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/7/smb/ : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":7},{"id":"436ecad829ea96713c74c75f411ba636b5f44868","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-24 08:20:23.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/7/check/tempest-dsvm-full-bionic/63b61f1/ : FAILURE in 1h 22m 22s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/7/check/devstack-plugin-ceph-dsvm-bionic/bdf8739/ : FAILURE in 36m 51s","accounts_in_message":[],"_revision_number":7},{"id":"2a7a0978603c9c3e28354de97004dcaea7da237b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-24 08:39:36.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-729603-7/ns5_nfs : SUCCESS in 2h 8m 29s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-729603-7/ns5_iscsi : SUCCESS in 2h 8m 56s","accounts_in_message":[],"_revision_number":7},{"id":"5bbbd1d39e116c6590c4eba87250fbce2da2a323","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-24 09:43:43.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"c216b7418c5faa7799e0aca678df9647eb6a620c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-24 10:16:32.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"30792a629690073c4335e09ee165c6c05f5a6cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:08:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be990e357f75436cbea48e62d1a05dad : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123e690ef25b462d8da38fae4a2f0b7b : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4dbce4a44b14f26a05185c708fb9bb3 : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a48a3dff20e549b6b079cfbfbfed8e29 : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c090f72db842e3ada55a0ccdcb602e : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ec6edf354a8a448aa9de06f26a352a50 : SUCCESS in 44m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1f685c6cc3541c88aa05d9b76f8aecd : SUCCESS in 54m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47530c7f38f14cb8a8c3fb8f85769bb3 : SUCCESS in 11m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/53bd94b1f83f4f31aba2418eea26488b : SUCCESS in 11m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a03810b88a04351b3105e02fecd6e58 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/30d99d45c076493e908b2566d3a73bad : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3da90b84bfc04444b2ac7a1ce421428b : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/de739224fd5c464ebe7d0b4d4e49454b : SUCCESS in 43m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c7bc7a57375b4d0cb94f3521aac6915c : FAILURE in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ca379b6dc8fd4721b1349f1b4a7219f5 : SUCCESS in 1h 13m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b6af8437f5948efb669c15e9e9b156b : FAILURE in 1h 07m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a75a1e10b69f461e9810de9ed175bf7b : SUCCESS in 1h 40m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/35fd1af4bb0a4331b57e78296dd68ab1 : SUCCESS in 1h 18m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f1c6f705bdfe42e4be17e3c56b0711b0 : SUCCESS in 1h 21m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/27d4258d079b433d8c1b1efdd8714817 : SUCCESS in 1h 14m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e354f6417997476194984f82c5a2a606 : SUCCESS in 2h 00m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/804fb530f72e403fa7cda4b8071d79b6 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":7},{"id":"4124e12d271126fa7032f2ddff91f83a1a09392c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-24 12:15:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/729603/7/upstream-check/cinder-cDOT-iSCSI/54c4191/ : FAILURE in 5m 37s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/03/729603/7/upstream-check/cinder-cDOT-iSCSI-IPv6/6bece6c/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/729603/7/upstream-check/cinder-cDOT-NFS/8477b9d/ : FAILURE in 5m 30s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/03/729603/7/upstream-check/cinder-cDOT-NFS-IPv6/c96f9e5/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/729603/7/upstream-check/cinder-cDOT-FCP/4f38873/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"24190541af20388281d8fa2026f4b2b821ca8322","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-24 12:18:37.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/03/729603/7/check/synology-dsm-cinder-iscsi/1c35074 : FAILURE in 9m 18s","accounts_in_message":[],"_revision_number":7},{"id":"9938b331e46265308bebdc589f33a3619d98f430","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-24 12:32:09.000000000","message":"Patch Set 7:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-03-729603-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"b574086fe5b8ea92dc779bac75ddb7fa92f4b233","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-09-24 12:35:04.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/03/729603/7/check/DellEMC_PowerFlex/426ebd1/DellEMC_PowerFlex/540 : FAILURE in 26m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"5b26f21481fec5ca1bbbe9726359d58c814a2e01","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-24 12:36:11.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_VNX_iSCSI/d985dd6/DellEMC_VNX_iSCSI/468 : FAILURE in 27m 18s\n- DellEMC_VNX_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_VNX_FC/264a973/DellEMC_VNX_FC/481 : FAILURE in 26m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"bb948c5d180077edd288fee142eac175da3451bc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-24 13:07:31.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_FC/3c9d496/DellEMC_Unity_FC/742 : FAILURE in 37m 30s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_iSCSI/04e613d/DellEMC_Unity_iSCSI/708 : FAILURE in 31m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"4ebcf2dddc0cf0bfb14335babc14f11e012c5c4d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-24 13:09:06.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_PowerMAX_FC/fb4a821/DellEMC_PowerMAX_FC/732 : FAILURE in 37m 58s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_PowerMAX_iSCSI/3b10cf8/DellEMC_PowerMAX_iSCSI/645 : FAILURE in 28m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"28eab4cec8c84dceb180936dce32733536fbd8dc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-09-24 13:47:42.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_PowerStore_iSCSI/8e14836/DellEMC_PowerStore_iSCSI/278 : SUCCESS in 1h 00m 22s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_PowerStore_FC/068ee01/DellEMC_PowerStore_FC/277 : SUCCESS in 1h 06m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"43576054e17fd8455ca2f7da35583b75bd7bdcd2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-24 13:58:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/729603/7/upstream-solidfire-check/cinder-solidfire-iSCSI/f6331e3/ : FAILURE in 1h 48m 47s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"843ed0066f78f51e06a194622563d0607c44df13","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-24 15:26:28.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"8568196fa652837af85c971108c310a19f347fd6","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-25 00:00:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/7/20200925061828/NEC-ISCSI : SUCCESS in 2h 12m 23s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/7/20200925061828/NEC-FC : SUCCESS in 2h 41m 53s","accounts_in_message":[],"_revision_number":7},{"id":"f44459e0e1d2f46af7f948652a2937961d26cb1e","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-09-25 00:22:07.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/03/729603/7/check/nimble-iscsi-driver-dsvm-volume/f6a1794 : FAILURE in 2h 27m 56s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/03/729603/7/check/nimble-fc-driver-dsvm-volume/241db58 : ABORTED in 24m 29s","accounts_in_message":[],"_revision_number":7},{"id":"9970e3a7ceb95f9dbd739834ca5e90ca35539559","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-09-25 03:30:10.000000000","message":"Patch Set 7:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":7},{"id":"a5bb3ebfdf51e78bfc8dd4f4406720f587c2ac65","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-09-25 06:44:08.000000000","message":"Patch Set 7:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":7},{"id":"ada087dd9e1e0c724c25e83f4378f95ae581018c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-25 08:25:30.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_FC/505cc41/DellEMC_Unity_FC/758 : SUCCESS in 1h 14m 58s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_iSCSI/c2051f6/DellEMC_Unity_iSCSI/724 : FAILURE in 1h 41m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"9f5ef41c5aba3793924b7687b8c3f23ded56992e","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-09-25 08:27:37.000000000","message":"Patch Set 7:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":7},{"id":"8fdaff26d9318c651f7b30211559098b3f2cab9f","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-09-25 08:30:44.000000000","message":"Patch Set 7:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":7},{"id":"96f8b6baaad7ff520797a51eb111d694683b8613","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-09-25 08:31:53.000000000","message":"Patch Set 7:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":7},{"id":"214594737cee5725908def4d2d8a5a5603489a83","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-25 08:58:50.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"4b6bfc92ee75dedc372f6e217a26fb4fea71e731","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-25 10:13:25.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_FC/5511ba9/DellEMC_Unity_FC/762 : SUCCESS in 1h 25m 46s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_iSCSI/c063e84/DellEMC_Unity_iSCSI/727 : FAILURE in 1h 41m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"8e2a36b11d25b2ce411d469f238c050db97ff40d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-25 10:35:42.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"32440db0d4179e93f4069f651a8be67373ec81d0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-25 21:57:45.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_XtremIO_FC/e26cf56/DellEMC_XtremIO_FC/1001 : FAILURE in 35m 48s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_XtremIO_iSCSI/0685011/DellEMC_XtremIO_iSCSI/963 : SUCCESS in 1h 34m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"fd333504e8934643290835d9305c93deb0dbe8b2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-26 01:52:42.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"9620579d37fa18069ad13edc1e4cda0c64a80110","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-26 08:08:47.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/729603/7/check/cinder-storpool-tempest/0c0d638/ : RETRY_LIMIT in 5m 36s","accounts_in_message":[],"_revision_number":7},{"id":"5ee7389581bfd9d9e8eab86e74519e0d20e775ac","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-09-27 03:07:09.000000000","message":"Patch Set 7:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":7},{"id":"603a66fbaeeef55a25467d24453a7847facd2db5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-27 04:55:27.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_FC/a2567d0/DellEMC_Unity_FC/773 : SUCCESS in 1h 19m 42s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/7/check/DellEMC_Unity_iSCSI/eca684e/DellEMC_Unity_iSCSI/738 : SUCCESS in 1h 48m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"4c396e9b2f3b07a8e6ca1f278efe58d9ad1a0c6a","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-09-27 05:03:53.000000000","message":"Patch Set 7:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":7},{"id":"009f527d049f8fcf952024cf68c826935f5e9ea2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-27 05:30:53.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"b58c7fe66d0cbbc886bf56bc7ea7a4430b7c7bc1","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-09-28 18:36:21.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c7c22f678a270159f13b0a94fa92068fab80d48d","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2020-10-02 18:38:30.000000000","message":"Patch Set 7:\n\nrun-storpoolci","accounts_in_message":[],"_revision_number":7},{"id":"033b70f6f4317e3b4196a882defdeaa6cd338954","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-10-03 07:20:54.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/03/729603/7/check/dsvm-tempest-zadara-driver/b2a3a3c : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4a48d7264e4cb0b49caf364f784ee43d4dad29b8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-10-04 04:07:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/03/729603/7/check/cinder-storpool-tempest/8fb25bb/ : SUCCESS in 1h 25m 17s","accounts_in_message":[],"_revision_number":7},{"id":"9255af67e8e6c804ccdb59ba082ea98489c2a112","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-11-16 02:50:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d755679038649bbe2529bbc2b1241708efb80ca7","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 03:03:46.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/1a27f837603c4fbe1e697315574118113df05ea2/index.html : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"f1f7183fdada081347bd1ff5893115fb1777a2a4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-16 03:12:33.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/03/729603/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4d707cc : FAILURE in 21m 18s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/03/729603/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/43df4c0 : FAILURE in 19m 57s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"84f9f8f5a6b91acb65c7a62e2d60bbf995ac8e7b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-16 03:26:43.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"b90d0ce79ec8fd8a85510cc4ffb2c6f3720ebbad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-16 03:30:06.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/03/729603/8/check-not-vote/ext-cinder-zuul/8fa3290 : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":8},{"id":"454c429ba59addc2ef3cb8e9253dbc2e6986b39b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-16 03:54:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/8/iscsi/ : SUCCESS in 1h 03m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/8/smb/ : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":8},{"id":"d6843cca44dd435b52076be364170bb207806d67","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-11-16 04:02:42.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/03/729603/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/91db27a/91db27a.html : SUCCESS in 1h 09m 55s","accounts_in_message":[],"_revision_number":8},{"id":"0595a99f75ed08e80b998c695f6246b5e8bfe2cf","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-11-16 04:08:50.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/9d0fe2d9fa3d44cab6910fcc713ea5eb : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":8},{"id":"4e5da0d19799036731acaf4ac08b1649e9d388e3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-16 04:14:42.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/8/check/tempest-dsvm-full-focal-py3/eba10d6/ : SUCCESS in 1h 19m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/8/check/devstack-plugin-ceph-dsvm-focal-py3/ad5ad2f/ : SUCCESS in 1h 18m 55s","accounts_in_message":[],"_revision_number":8},{"id":"33332358334abd9881d4e44d8623258125a7e957","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-16 04:23:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/8/check/dsvm-tempest-cisco-zonemanager-job/c670ea1/ : FAILURE in 1h 28m 38s","accounts_in_message":[],"_revision_number":8},{"id":"b777914c1c7321a6c7a6f57760913d4d0ceb8a19","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 04:31:08.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_XtremIO_FC/24eaddf/DellEMC_XtremIO_FC/86 : FAILURE in 1h 39m 56s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_XtremIO_iSCSI/9fb678d/DellEMC_XtremIO_iSCSI/84 : SUCCESS in 1h 38m 41s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"9c16e08e93eb704f48d5a7a5dd242fff5b05d545","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-16 04:50:28.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4ccbcc88e2aa391442e1f00143bd8bac6bc740ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 04:58:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc19eab07aa84f3b8ec70cf021128151 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02e8e5e4cff41258e46ead989e57d38 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01fe2a2982ce42dbacc025e88f475479 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff430d28baf84abfa8f5b567f51a7c0b : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/484c40973727401ebb4a65ee6ab664e8 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3218748d590246d980c23b609f899da4 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/a2daa7b95a834e69b0612ea0bd29afad : SUCCESS in 51m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fef3cdc025404158bf9ae5916c38eb37 : SUCCESS in 1h 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37c858c39a014244a15a89dc1aee2a90 : SUCCESS in 5m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e9d693bf83964173a267426d9918d8a0 : SUCCESS in 11m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cb346d805871421690220deac0c73dce : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7405a9c577447a18b54598312a68dab : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3cda2f75bac942c4aa1af5178ca8a572 : SUCCESS in 4m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a944235ed0b04606b1a7140c62aa486a : SUCCESS in 44m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37d84de0f6b64998be7201c8832cce78 : FAILURE in 4m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a385059f9a1449a999c6633aa338fe4 : SUCCESS in 1h 09m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d7a4831cee1b46e6ab1f6cadaa0ed923 : FAILURE in 1h 06m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/67f7be6446a34e8dbae0f5ab4ff5003d : SUCCESS in 1h 34m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1fb8ad99d3fd40f89bdb8ce37ccac67d : SUCCESS in 1h 01m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/242ec63f1ef84042ae2bf83830df8e32 : FAILURE in 1h 19m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/74b5e8ee15004daf95dbd2704cb566a3 : SUCCESS in 1h 25m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/daee7a3733c943899e9a50384634a1ae : FAILURE in 2h 06m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42bc3f11c320454ca3eab1177e112077 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":8},{"id":"69cf43717363e424ef973d3c222664c6ab2be2f7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-16 05:49:49.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-729603-8/ns5_nfs : SUCCESS in 2h 58m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-03-729603-8/ns5_iscsi : SUCCESS in 2h 51m 15s","accounts_in_message":[],"_revision_number":8},{"id":"01db9d70b11a9db09cfce4bab5078f31397ad448","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-11-16 05:52:17.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"1254f9c5ae182598ce9eabb9503314e09117bb08","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-16 06:33:51.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/03/729603/8/check-not-vote/ext-cinder-zuul/9431641 : SUCCESS in 40m 52s","accounts_in_message":[],"_revision_number":8},{"id":"b8a0c0149f7ebc82a420d872f5978ad27ab68bbf","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-16 06:36:47.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/03/729603/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/8c1302a : FAILURE in 21m 28s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/03/729603/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6ab7448 : FAILURE in 20m 58s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"8896b0f88d3f639abedfe20e022889e51932d55f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-16 06:45:32.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"6e9b1ef4688b699158e7227f46e20fab3c55e0b8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-16 07:09:54.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/729603/8/iscsi/ : SUCCESS in 1h 17m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/729603/8/smb/ : FAILURE in 1h 04m 05s","accounts_in_message":[],"_revision_number":8},{"id":"d3a821e5b0c489a2fce77a93d4ac80a6e8f3b327","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-16 07:10:30.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/03/729603/8/check/dsvm-tempest-cisco-zonemanager-job/2bcd45a/ : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":8},{"id":"fdbd76232253a4baa5a2228b3b50b86ad775bcd0","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 07:54:45.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_PowerMAX_FC/e6738f3/DellEMC_PowerMAX_FC/117 : SUCCESS in 4h 35m 25s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_PowerMAX_iSCSI/87d05c6/DellEMC_PowerMAX_iSCSI/108 : SUCCESS in 5h 03m 38s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"4dcfc6c0dbd3742adb5e8db13b43f70825a24ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 07:55:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40c041654af443e8bdf90245476a10c6 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f8d96f43444dce9db2aa53b8c26881 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1a6ec7a5f0a4888814dbf684e898268 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33f0298b6dbb461c9de9bb931c70c33b : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e49ced161ff47f78fc66c6dce285239 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed4701d650e4f8687244c5db544f829 : SUCCESS in 15m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/e0f685f0ab7a45be94875fa1135158d7 : SUCCESS in 54m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/daa8f0ec5c2e445dac334236daa25437 : SUCCESS in 1h 08m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36e980447f3a4587b5cf20e4567faf14 : SUCCESS in 6m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/faeba4478200427db5846945184c1f14 : SUCCESS in 11m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f704cc917a1349a49ef0006c167c0a9f : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a522be748d841cd91839321af42f4c3 : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40100727e7a04df395db715c9902b182 : SUCCESS in 5m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f2abc6ffba874123a3f7b1d88b04c91c : SUCCESS in 41m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6d1acd48074348ccb0cdf6e586c60e2b : FAILURE in 4m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e736e7fbeaad446b807c9aeb1729f8d4 : SUCCESS in 1h 02m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8b9bae5e3bfc4d9ea1eb95d26e772c26 : FAILURE in 1h 12m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/23c25ea2a6764c25b61f9ef31a681ecf : SUCCESS in 1h 22m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/023fd56c961f4cfda381d75410b3a3fc : SUCCESS in 1h 10m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/125445fe5c7e45aca921978b7546a081 : SUCCESS in 1h 20m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/393485424e174fcf8c5105d01190473b : SUCCESS in 1h 13m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/101d5ad2da5044038285e3098d553fe4 : SUCCESS in 1h 59m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/399362f83e164d2c9a2799d6f064aa7e : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":8},{"id":"8b1a652990727d40fae29b059a80c09b9a605349","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-16 08:06:59.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"5936b9937169855384fb925afe6d049aa64506c0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-16 08:35:37.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/03/729603/8/upstream-check/cinder-cDOT-iSCSI/b493d1b/ : FAILURE in 32m 39s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/03/729603/8/upstream-check/cinder-cDOT-iSCSI-IPv6/61dac40/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/03/729603/8/upstream-check/cinder-cDOT-NFS/2eba0b6/ : FAILURE in 39m 34s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/03/729603/8/upstream-check/cinder-cDOT-NFS-IPv6/b435b59/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/03/729603/8/upstream-check/cinder-cDOT-FCP/2beef2b/ : FAILURE in 7m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"053e0c0455bdd4719a3dd38afee7c2ed026b9470","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-16 08:43:43.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/8/check/tempest-dsvm-full-focal-py3/c7576b4/ : SUCCESS in 1h 12m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/729603/8/check/devstack-plugin-ceph-dsvm-focal-py3/98a0fd4/ : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":8},{"id":"d5eaccc81a7c77f20cda4108bee2c82ec5465c35","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-11-16 09:32:32.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/03/729603/8/check/synology-dsm-cinder-iscsi/2ca5c30 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":8},{"id":"b87fb2ff95b1cac1c999c9bfb3102cb1e8a03ede","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-11-16 09:37:43.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_PowerStore_iSCSI/45e56a2/DellEMC_PowerStore_iSCSI/92 : SUCCESS in 1h 04m 00s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_PowerStore_FC/92187ee/DellEMC_PowerStore_FC/88 : SUCCESS in 1h 04m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"69a69df2f638a825f2bd127fd15d4decfa8758f5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-11-16 09:40:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/03/729603/8/upstream-solidfire-check/cinder-solidfire-iSCSI/936cdce/ : SUCCESS in 1h 44m 30s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"08fe769246a23d3dc707968a0cf40fac3e15890a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-16 10:39:23.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"d4e5850ffdaffce88ddf7baaebbc099753dc9438","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-11-16 11:37:07.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/729603-8/Seagate-iSCSI : SUCCESS in 44:17.53\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/729603-8/Seagate-FC : SUCCESS in 40:42.45","accounts_in_message":[],"_revision_number":8},{"id":"eadd07b0f82081209d5c5655d74f50a3ebe468fd","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-11-16 11:41:55.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/729603-8/Lenovo-iSCSI : SUCCESS in 49:27.98\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/729603-8/Lenovo-FC : SUCCESS in 43:24.36","accounts_in_message":[],"_revision_number":8},{"id":"c10bebb5fce07d5fa61a0af2fe562db34366c0f7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 13:16:55.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_VNX_iSCSI/5223abe/DellEMC_VNX_iSCSI/84 : SUCCESS in 2h 46m 45s\n- DellEMC_VNX_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_VNX_FC/1a14a41/DellEMC_VNX_FC/85 : SUCCESS in 1h 12m 24s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"de9690b2df446dec1be2a2cdc34be2df2e67f85d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-16 13:19:37.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"11ca03deec241281f02aa4c8e0304b9d66367741","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 13:53:47.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_XtremIO_FC/ce47c7a/DellEMC_XtremIO_FC/91 : SUCCESS in 1h 40m 31s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_XtremIO_iSCSI/3f49c07/DellEMC_XtremIO_iSCSI/89 : SUCCESS in 1h 35m 54s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"849a403d5525a2fa1f78c79399fca1b8a8e8f316","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-16 14:40:18.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"b9a2f81873e07bceb9b6d3f2410afdafc6b807f8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 15:52:52.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/8/20201116224648/NEC-ISCSI : FAILURE in 1h 53m 45s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/03/729603/8/20201116224648/NEC-FC : FAILURE in 2h 06m 02s","accounts_in_message":[],"_revision_number":8},{"id":"b4ecf7a97eb71a759ed3b232a32fc0a081175060","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-16 16:55:09.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"070967bd7a98e1f232d3b74af9a6505641d08860","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-11-17 06:41:10.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_PowerMAX_FC/83fc0fb/DellEMC_PowerMAX_FC/122 : SUCCESS in 4h 17m 05s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_PowerMAX_iSCSI/ec4a720/DellEMC_PowerMAX_iSCSI/113 : SUCCESS in 4h 22m 27s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"ce04b54bf2e5489dceba48393e28fa8688b55a22","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-17 07:26:17.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/03/729603/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"95c8632bb4d0b82ab4656b8e2e3717df6d907931","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-11-17 17:32:58.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpemsa-ci/729603-8/HPMSA-iSCSI : SUCCESS in 53:47.60\n* HPMSA-FC http://os-logs.tristero.net/hpemsa-ci/729603-8/HPMSA-FC : SUCCESS in 37:29.37","accounts_in_message":[],"_revision_number":8},{"id":"f1cb8b08f5e180b5b657bb0cd774fbb580b78d98","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-11-18 11:26:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3f9ed2f6f51c89d8297cf9f55a97a182aeb02c40","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2021-02-01 06:50:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ab33a4b1fd1b621b79eda29d0ece53ddae380eb5","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-02-07 09:38:38.000000000","message":"Patch Set 8:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":8},{"id":"83516debec0ba3f10d82e3ea9c5f3875fb45df66","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-07 15:06:36.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_Unity_FC/12c7d97/DellEMC_Unity_FC/616 : FAILURE in 1h 36m 58s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_Unity_iSCSI/2b6cc6c/DellEMC_Unity_iSCSI/581 : FAILURE in 2h 06m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"53558e2b3195c2042f18cf7297898c13f757a41f","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-02-08 02:08:07.000000000","message":"Patch Set 8:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":8},{"id":"187683c935f1ae3303e65f53fa6c562c44e1fea0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-08 04:08:33.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_Unity_FC/d627fe7/DellEMC_Unity_FC/618 : FAILURE in 1h 59m 15s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_Unity_iSCSI/2311378/DellEMC_Unity_iSCSI/583 : SUCCESS in 1h 59m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"cbfef5ff43e4e2b4649625262dac205632c2100b","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-02-08 04:14:59.000000000","message":"Patch Set 8:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":8},{"id":"6acbcd856bcd3e5f69443e9b46ca2908a67178ae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-08 06:00:54.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_Unity_FC/6fe99b6/DellEMC_Unity_FC/619 : FAILURE in 1h 33m 15s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_Unity_iSCSI/0af7d48/DellEMC_Unity_iSCSI/584 : SUCCESS in 1h 45m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"623d3154a3c9395f6888c499861e5d7d2cc543b1","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-02-08 06:25:35.000000000","message":"Patch Set 8:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":8},{"id":"bacff3a965911d8dbaa94c94abc652850b0f271c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-08 08:02:41.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/03/729603/8/check/DellEMC_Unity_FC/cf7989d/DellEMC_Unity_FC/620 : SUCCESS in 1h 31m 25s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/03/729603/8/check/DellEMC_Unity_iSCSI/e07e7c7/DellEMC_Unity_iSCSI/585 : SUCCESS in 1h 36m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"89457c02c7aeea73fed02df59a9c5b27c981c6d6","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-02-18 09:52:01.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"c935721babe839e5dd0f1e64c0a2e38b7e3eb44d","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-16 07:03:17.000000000","message":"Patch Set 8: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"287a63efd76b2397d8d21c55dae292887d6f475b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-09-17 15:22:31.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nQuestion inline.  Also, probably worth rebasing on master so we can get updated test results.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1a27f837603c4fbe1e697315574118113df05ea2","revisions":{"68c9b0ab03b2cf4be62f31714ddcf2a33309c44c":{"kind":"REWORK","_number":1,"created":"2020-05-20 13:43:40.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/03/729603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/729603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/729603/1"}}},"commit":{"parents":[{"commit":"85e60732e21ee30937d43d8b77295e4525e1c8fc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e60732e21ee30937d43d8b77295e4525e1c8fc"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:31.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:34.000000000","tz":0},"subject":"[Unity]: delete lun with hosts accessed","message":"[Unity]: delete lun with hosts accessed\n\nIf lun in Unity backend has hosts accessed, it can\u0027t be deleted.\nIt will cause volume failed to delete in OpenStack.\nSo remove host access before lun deletion.\n\nChange-Id: I0887f48c28741f434c6a48041c87e849e471bb94\nCloses-bug: #1879705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68c9b0ab03b2cf4be62f31714ddcf2a33309c44c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68c9b0ab03b2cf4be62f31714ddcf2a33309c44c"}]},"branch":"refs/heads/master"},"cba14e50e6981feef6e6b71410990f353bb70229":{"kind":"REWORK","_number":2,"created":"2020-05-20 13:51:42.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/03/729603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/729603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/729603/2"}}},"commit":{"parents":[{"commit":"85e60732e21ee30937d43d8b77295e4525e1c8fc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e60732e21ee30937d43d8b77295e4525e1c8fc"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:31.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:51:28.000000000","tz":0},"subject":"[Unity]: delete lun with hosts accessed","message":"[Unity]: delete lun with hosts accessed\n\nIf lun in Unity backend has hosts accessed, it can\u0027t be deleted.\nIt will cause volume failed to delete in OpenStack.\nSo remove host access before lun deletion.\n\nChange-Id: I0887f48c28741f434c6a48041c87e849e471bb94\nCloses-bug: #1879705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cba14e50e6981feef6e6b71410990f353bb70229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cba14e50e6981feef6e6b71410990f353bb70229"}]},"branch":"refs/heads/master"},"bf8c9aa965ebbcc6ba94a8f9ca18b031d6efaada":{"kind":"REWORK","_number":3,"created":"2020-06-01 02:41:48.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/03/729603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/729603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/729603/3"}}},"commit":{"parents":[{"commit":"85e60732e21ee30937d43d8b77295e4525e1c8fc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e60732e21ee30937d43d8b77295e4525e1c8fc"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:31.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-06-01 02:41:52.000000000","tz":0},"subject":"[Unity]: delete lun with hosts accessed","message":"[Unity]: delete lun with hosts accessed\n\nIf lun in Unity backend has hosts accessed, it can\u0027t be deleted.\nIt will cause volume failed to delete in OpenStack.\nSo remove host access before lun deletion.\n\nChange-Id: I0887f48c28741f434c6a48041c87e849e471bb94\nCloses-bug: #1879705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf8c9aa965ebbcc6ba94a8f9ca18b031d6efaada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf8c9aa965ebbcc6ba94a8f9ca18b031d6efaada"}]},"branch":"refs/heads/master"},"ead6bf08203f37c309043e43f8edbe2736ebacc9":{"kind":"REWORK","_number":4,"created":"2020-06-01 03:27:45.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/03/729603/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/729603/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/729603/4"}}},"commit":{"parents":[{"commit":"85e60732e21ee30937d43d8b77295e4525e1c8fc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e60732e21ee30937d43d8b77295e4525e1c8fc"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:31.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-06-01 03:27:55.000000000","tz":0},"subject":"[Unity]: delete lun with hosts accessed","message":"[Unity]: delete lun with hosts accessed\n\nIf lun in Unity backend has hosts accessed, it can\u0027t be deleted.\nIt will cause volume failed to delete in OpenStack.\nSo remove host access before lun deletion.\n\nChange-Id: I0887f48c28741f434c6a48041c87e849e471bb94\nCloses-bug: #1879705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ead6bf08203f37c309043e43f8edbe2736ebacc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ead6bf08203f37c309043e43f8edbe2736ebacc9"}]},"branch":"refs/heads/master"},"809d33a0c2395a4448f13863648c418aca7dc72d":{"kind":"REWORK","_number":5,"created":"2020-06-05 02:42:45.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/03/729603/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/729603/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/729603/5"}}},"commit":{"parents":[{"commit":"85e60732e21ee30937d43d8b77295e4525e1c8fc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e60732e21ee30937d43d8b77295e4525e1c8fc"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:31.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-06-05 02:38:13.000000000","tz":480},"subject":"[Unity]: delete lun with hosts accessed","message":"[Unity]: delete lun with hosts accessed\n\nIf lun in Unity backend has hosts accessed, it can\u0027t be deleted.\nIt will cause volume failed to delete in OpenStack.\nSo remove host access before lun deletion.\n\nChange-Id: I0887f48c28741f434c6a48041c87e849e471bb94\nCloses-bug: #1879705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/809d33a0c2395a4448f13863648c418aca7dc72d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/809d33a0c2395a4448f13863648c418aca7dc72d"}]},"branch":"refs/heads/master"},"3ff06ddd125e20369c6212e49b3a47d9de453aac":{"kind":"REWORK","_number":6,"created":"2020-06-12 05:39:42.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/03/729603/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/729603/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/729603/6"}}},"commit":{"parents":[{"commit":"85e60732e21ee30937d43d8b77295e4525e1c8fc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e60732e21ee30937d43d8b77295e4525e1c8fc"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:31.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-06-12 05:36:58.000000000","tz":480},"subject":"[Unity]: delete lun with hosts accessed","message":"[Unity]: delete lun with hosts accessed\n\nIf lun in Unity backend has hosts accessed, it can\u0027t be deleted.\nIt will cause volume failed to delete in OpenStack.\nSo remove host access before lun deletion.\n\nChange-Id: I0887f48c28741f434c6a48041c87e849e471bb94\nCloses-bug: #1879705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ff06ddd125e20369c6212e49b3a47d9de453aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ff06ddd125e20369c6212e49b3a47d9de453aac"}]},"branch":"refs/heads/master"},"105c0855fa12da0ad2878261a136b1c7b0c830c9":{"kind":"REWORK","_number":7,"created":"2020-09-24 06:30:16.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/03/729603/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/729603/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/729603/7"}}},"commit":{"parents":[{"commit":"85e60732e21ee30937d43d8b77295e4525e1c8fc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e60732e21ee30937d43d8b77295e4525e1c8fc"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:31.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-09-24 06:27:54.000000000","tz":480},"subject":"[Unity]: delete lun with hosts accessed","message":"[Unity]: delete lun with hosts accessed\n\nIf lun in Unity backend has hosts accessed, it can\u0027t be deleted.\nIt will cause volume failed to delete in OpenStack.\nSo remove host access before lun deletion.\n\nChange-Id: I0887f48c28741f434c6a48041c87e849e471bb94\nCloses-bug: #1879705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/105c0855fa12da0ad2878261a136b1c7b0c830c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/105c0855fa12da0ad2878261a136b1c7b0c830c9"}]},"branch":"refs/heads/master"},"1a27f837603c4fbe1e697315574118113df05ea2":{"kind":"REWORK","_number":8,"created":"2020-11-16 02:50:50.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/03/729603/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/729603/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/729603/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/729603/8"}}},"commit":{"parents":[{"commit":"85e60732e21ee30937d43d8b77295e4525e1c8fc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e60732e21ee30937d43d8b77295e4525e1c8fc"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-05-20 13:40:31.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-11-16 02:45:55.000000000","tz":480},"subject":"[Unity]: delete lun with hosts accessed","message":"[Unity]: delete lun with hosts accessed\n\nIf lun in Unity backend has hosts accessed, it can\u0027t be deleted.\nIt will cause volume failed to delete in OpenStack.\nSo remove host access before lun deletion.\n\nChange-Id: I0887f48c28741f434c6a48041c87e849e471bb94\nCloses-bug: #1879705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a27f837603c4fbe1e697315574118113df05ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a27f837603c4fbe1e697315574118113df05ea2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
