)]}'
{"id":"openstack%2Fcinder~732400","triplet_id":"openstack%2Fcinder~stable%2Ftrain~I66da0212a61d778fe9795d382b3af5b225a8aa54","project":"openstack/cinder","branch":"stable/train","topic":"bug/1875570","hashtags":[],"change_id":"I66da0212a61d778fe9795d382b3af5b225a8aa54","subject":"Creating image-volume cache on NFS backend fails","status":"MERGED","created":"2020-06-01 14:59:39.000000000","updated":"2020-08-21 15:04:54.000000000","submitted":"2020-08-21 14:40:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732400-1598020826413-3ae6a7cc","meta_rev_id":"ab2b69dc571a3047ff92047c32adfe5ebe2bce01","_number":732400,"virtual_id_number":732400,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-08-21 00:28:07.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-08-21 02:21:44.000000000","_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2020-08-20 23:28:51.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-08-20 13:52:40.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-20 19:55:24.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 14:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"date":"2020-08-20 21:03:19.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-20 18:44:20.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-21 08:08:39.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-08-20 23:47:03.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-20 22:56:58.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-08-21 09:56:00.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"date":"2020-08-20 18:39:04.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":2,"date":"2020-08-20 13:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":2,"date":"2020-08-21 13:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":1,"date":"2020-08-21 13:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 14:59:39.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2020-06-01 15:00:45.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-06-01 15:29:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-06-01 17:36:39.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-01 21:36:57.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-20 09:17:11.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-20 11:24:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-20 13:14:22.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-08-20 13:52:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-20 18:44:20.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-20 19:55:24.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-08-20 21:03:19.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-20 22:56:58.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-08-20 23:28:51.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-20 23:47:03.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 00:28:07.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-08-21 02:21:44.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-08-21 08:08:39.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-08-21 09:56:00.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 13:00:56.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-08-21 14:40:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9585812019987fbbc292794f4f01194b35d16db","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-01 14:59:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"5afd18af0278f7daea938ed6fc779f9ab9a3460e","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-01 15:00:45.000000000","message":"Patch Set 1: Code-Review+2\n\nGood backport.  Makes sense to merge this back to more branches.","accounts_in_message":[],"_revision_number":1},{"id":"85b694498262f687565bd6babd0396dccda75209","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 15:10:56.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/075fe1ff80ed9509d4e8ddd41f8670c72d5567d1/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f95d6f792fe6b8b574c9da43aaf9403b1e3ad7ad","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-01 15:29:34.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/00/732400/1/check/dsvm-tempest-infinibox-fc/b5bd699 : FAILURE in 29m 13s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/00/732400/1/check/dsvm-tempest-infinibox-iscsi/524b637 : FAILURE in 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"821329e81f5efd9ad844891859fa0c391db967ae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-01 15:41:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/00/732400/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2c20885/2c20885.html : FAILURE in 39m 42s","accounts_in_message":[],"_revision_number":1},{"id":"90fcf2fc9f2dfe209bff9c2522151b2a5fcc1d8b","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-06-01 15:44:56.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/00/732400/1/check/Cinder-AS13000/1af5663 : FAILURE in 9m 38s\n- cinder-test-fc http://218.29.111.78:9000/00/732400/1/check/cinder-test-fc/19f254f : FAILURE in 39m 37s\n- cinder-test-iscsi http://218.29.111.78:9000/00/732400/1/check/cinder-test-iscsi/24a407d : FAILURE in 39m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eb50efe0e2ffe20b84cd37adc227d1e8d988db9f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-01 16:05:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/fe5ea30 : SUCCESS in 1h 01m 16s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/57ab296 : SUCCESS in 1h 05m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"c09d094d1ea937bd4dbacb036571990ae3e31151","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-01 16:08:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/732400/1/check/dsvm-tempest-cisco-zonemanager-job/49d5405/ : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":1},{"id":"85e21aeab58bd0c7fc333f02e7541c09bf296bae","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-01 16:10:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/1/iscsi/ : FAILURE in 1h 10m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/1/smb/ : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e703d6274da72c966d27633a3ccc9dfd80949b25","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-01 16:26:02.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/1/check/tempest-dsvm-full-bionic/f4ee39c/ : FAILURE in 21m 37s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/1/check/devstack-plugin-ceph-dsvm-bionic/f3c31b3/ : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dfd2c9df6d655297d034514f52016863009c0629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 17:33:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/596961bb13824142bb41e05d76c4a8e8 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad38e2721b140ad815e844a9af7ea36 : SUCCESS in 13m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8580a51732c4ea3996f6918b3b4c956 : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c786871d78148b183be37b9a8436f82 : SUCCESS in 14m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ae3960d84234c8188c8661742858a2c : SUCCESS in 16m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc53a89ac4854dfa8e2fd9a791c340b5 : SUCCESS in 17m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d9c1534bd3f344198fb99f9378c2c619 : FAILURE in 1h 28m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/05b361c3549041a6b42b9e8e15ca6657 : SUCCESS in 55m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ae64b9b3d964b6a84be13d7e352c198 : SUCCESS in 1h 43m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e1400e0ab214549be06beab292b7745 : SUCCESS in 10m 24s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02498546f1d54e97a57ea8e08eac0fbd : SUCCESS in 1h 39m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9314d7bcad64c808875a30894cc7d8e : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76bf28fb19c941edb813b005c8afb2a6 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2abc098f990848e9814b1ba7afee7535 : FAILURE in 16m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5275797ea4df47d89353b2d6c8b3bee2 : SUCCESS in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d55f4cbbaa2b43ff99e2ebda5f5bc8ba : FAILURE in 1h 13m 51s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/52848884117c4ec0b449987b4a2b865b : SUCCESS in 6m 11s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/452c0067574b4fc3939bb4497a1e8b56 : SUCCESS in 1h 10m 10s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d725baa99e07439b998d0cc3a275e9a9 : FAILURE in 1h 50m 16s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/fba703fed2af4f8eadc0fbba31377b93 : SUCCESS in 1h 25m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a30e19904ac847d58e40f9efac52d10b : SUCCESS in 2h 31m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ac599da835d54acfaba34a91a6f2c194 : SUCCESS in 56m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2ce51a44e2a4c278618fe5254a54371 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":1},{"id":"441034a64f3475c637f28fe66a7cca190f1fb8c3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-06-01 17:36:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/732400/1/20200601235942/NEC-FC : SUCCESS in 2h 29m 44s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/732400/1/20200601235942/NEC-ISCSI : SUCCESS in 2h 36m 54s","accounts_in_message":[],"_revision_number":1},{"id":"521c9ec49e9e0a96c63a7761f49dc74628e8220d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-01 17:38:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82a7a1a91980d98af2f6ed050d9d5c929396cd5b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-01 17:47:03.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d768e50d3956c4513bbd8f5157aafa244c3d694c","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-06-01 18:38:45.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/00/732400/1/check/Cinder-AS13000/fd18dcc : FAILURE in 7m 51s\n- cinder-test-fc http://218.29.111.78:9000/00/732400/1/check/cinder-test-fc/a963596 : FAILURE in 38m 43s\n- cinder-test-iscsi http://218.29.111.78:9000/00/732400/1/check/cinder-test-iscsi/d4babf7 : FAILURE in 38m 22s","accounts_in_message":[],"_revision_number":1},{"id":"940369a5e2836940710605bddc84e82faeff822d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-01 18:42:58.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/1/iscsi/ : SUCCESS in 1h 03m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/1/smb/ : FAILURE in 1h 00m 29s","accounts_in_message":[],"_revision_number":1},{"id":"de20b145f9118b579b8895d37bcf9bd373efb07c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-01 18:46:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/732400/1/check/dsvm-tempest-cisco-zonemanager-job/bd5b53a/ : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2198339b86c8a5e468ce6fc8a44e7b832d38e1c8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-01 19:12:02.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/1/check/tempest-dsvm-full-bionic/60626d5/ : FAILURE in 18m 57s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/1/check/devstack-plugin-ceph-dsvm-bionic/94fdc71/ : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2ec8881547c7b6e30304bb0e9c1acb37165d96a8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-01 19:19:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"451b22ee8d49058c46919202ff147b525cc326aa","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-01 19:49:20.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f993e410a11220d4a1a6ce21ae6601ea7fd4027d","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-06-01 19:56:09.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/00/732400/1/check/Cinder-AS13000/017869f : FAILURE in 7m 47s\n- cinder-test-fc http://218.29.111.78:9000/00/732400/1/check/cinder-test-fc/27c7def : FAILURE in 36m 21s\n- cinder-test-iscsi http://218.29.111.78:9000/00/732400/1/check/cinder-test-iscsi/91023b9 : FAILURE in 36m 12s","accounts_in_message":[],"_revision_number":1},{"id":"95d26ab173470c28d5d53e471a5a0a097cc8675a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-01 20:22:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/1/iscsi/ : SUCCESS in 1h 03m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/1/smb/ : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a4fa769f7eab0b513b23e9e833b186116d226b4d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-01 20:26:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/732400/1/check/dsvm-tempest-cisco-zonemanager-job/9349cbc/ : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"09fb3dcd1949d231d60ce68da77fbacab15fc2a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 20:30:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84326abd22454e35a6314d17e29e0f96 : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba52e05aa994ed6af5638ad4e681110 : SUCCESS in 11m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ef668f10a764fa18feb4199047f7937 : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b8c460d9ef340a8b27b79e0adb17e95 : SUCCESS in 9m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b5dda5f06a1461185c9c00fe1489e5c : SUCCESS in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e4196a679c496f9951b43ad6fe44e8 : SUCCESS in 12m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2fbc801d8a9347b3a50935084c57287e : SUCCESS in 1h 39m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f744ded3b4b145a08e589ee2b641b813 : SUCCESS in 1h 08m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16e0782d8d56442da293c0cea456e5b0 : SUCCESS in 1h 19m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e5cc2b64d8c4f9c83c022986d13876d : SUCCESS in 7m 01s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9fb5bfd8020d43ef9e8e3964f4598508 : SUCCESS in 1h 12m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e342df04afe54662800472b6de432a32 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74415f6ccfa9437ea0c835064b9fb7cf : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c522f47ccdc4d6daf4d040b486e2c62 : FAILURE in 18m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81bcf83d9da04382b2a2afafd87a36dc : SUCCESS in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c932517a867431883af2f0753e23870 : FAILURE in 1h 06m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a1f6df70d3074af1a3f2f013410d62ed : SUCCESS in 9m 06s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5ca1959a07be49eaaf739f310b3454eb : SUCCESS in 1h 10m 29s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ae20866ace942429e73c531d5f6465f : SUCCESS in 1h 26m 55s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/a8ba6e6000da454e8c496fe0bec40f48 : SUCCESS in 1h 26m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7dbd7af30c75470797b94b3b36bc5443 : FAILURE in 2h 48m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fbf0395e196348bdb4123c2da5a8a383 : SUCCESS in 1h 04m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af31b13a7402454288b1fb725fabe8eb : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9bcf9788075e88141149b77e32870da0431a6dd9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-01 20:35:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45d43d7f1a95e44e5a396d127be312e8e3f9d071","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-01 20:45:52.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/1/check/tempest-dsvm-full-bionic/ea9569e/ : FAILURE in 19m 53s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/1/check/devstack-plugin-ceph-dsvm-bionic/c70b21b/ : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":1},{"id":"75f6db257727816ae05442c1483e6e4366ef035a","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-06-01 21:36:57.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/00/732400/1/check/Cinder-AS13000/b5f69ef : FAILURE in 8m 05s\n- cinder-test-fc http://218.29.111.78:9000/00/732400/1/check/cinder-test-fc/14a7c48 : FAILURE in 34m 22s\n- cinder-test-iscsi http://218.29.111.78:9000/00/732400/1/check/cinder-test-iscsi/c4b161f : FAILURE in 34m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f5646dfc2fc773a2c7e7e98acccffce3a0882a48","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-01 21:39:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/1/iscsi/ : SUCCESS in 1h 00m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/1/smb/ : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fe65d80032f3e306621805c445ddbe034db4ae65","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-01 21:45:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/732400/1/check/dsvm-tempest-cisco-zonemanager-job/1244935/ : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":1},{"id":"be8a2f7821cd7e82ec4a2da68577573e8e88627c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-01 22:32:38.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1491b12cf62073a9eac72e5b5e71371496d9f8b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 22:46:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/215e54d749df46b59245228489fbb9aa : SUCCESS in 14m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b22b7d0229bd46dabb11c1fd6c6b9cc3 : SUCCESS in 9m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b8503c75d234d6eab3ff13da799e48c : SUCCESS in 9m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5619684b471144e6bd46965feeed9b6d : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1af759b2eef24fef80f2cf31a2824da4 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a4ef3668a74d4c92068f6a261907c3 : SUCCESS in 11m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5261bcadd53a4520b02e9f300cc4c309 : SUCCESS in 1h 33m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b9f7023cf3db45e0a28535086f740590 : SUCCESS in 1h 00m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd3f40807b8c476b9e6976e7d8a6939a : SUCCESS in 1h 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb2a388813154e858eff55913a705604 : SUCCESS in 8m 01s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/588809151c1748a08f8834fdd19f3b96 : SUCCESS in 1h 49m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d46cb23a19844a1089a39a77c5c524ce : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6da89ad875204e2b97689c483d86e975 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bae9ac4e767746fbaa3c0981934d07ee : FAILURE in 20m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8a5d630e30a4b619da283d2049f543a : SUCCESS in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/472bc4abcb6947bb8ddc771247595f23 : FAILURE in 1h 09m 14s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e88d21f79dab4300bf6c7f016f339683 : SUCCESS in 5m 56s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc2e13480d1c46c0a0b0f2cf618e6770 : SUCCESS in 1h 10m 37s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2d9b8e0a3efe4c109889063da7935527 : SUCCESS in 1h 24m 19s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/d68bf8152d5e48278c766957c4685779 : SUCCESS in 1h 31m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7230281810e84b8199c0e1c279d27736 : SUCCESS in 2h 04m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1f86b10e2ff944b1a22261af4bbf64d8 : SUCCESS in 1h 03m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/054c1ec8d91a4071aaeec75c615f1d37 : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":1},{"id":"522da686a13cc2915591917955457b22eed27bcb","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-01 22:49:02.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/a562b56 : SUCCESS in 1h 02m 17s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6da7ccc : FAILURE in 1h 51m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2f809345d17f2f545803e2e2781d9dc2663867b","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-06-01 22:52:24.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/00/732400/1/check/synology-dsm-cinder-iscsi/fed7021 : FAILURE in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"752216228b6cd7dbdc253f74f5ca96f72cc08216","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-01 23:53:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/00/732400/1/upstream-solidfire-check/cinder-solidfire-iSCSI/b3d0633/ : SUCCESS in 1h 05m 33s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"26babfcacc6f9d58b16e11e72a6af036c22cdbaa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-02 00:48:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/00/732400/1/upstream-check/cinder-cDOT-iSCSI/adbf01b/ : SUCCESS in 52m 01s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/00/732400/1/upstream-check/cinder-cDOT-iSCSI-IPv6/7a126d5/ : SUCCESS in 1h 26m 17s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/00/732400/1/upstream-check/cinder-cDOT-NFS/6b4232d/ : SUCCESS in 51m 16s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/00/732400/1/upstream-check/cinder-cDOT-NFS-IPv6/8b38c23/ : SUCCESS in 1h 28m 54s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/00/732400/1/upstream-check/cinder-cDOT-FCP/4d88929/ : FAILURE in 2h 01m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"eead3bca0ce12d195162a377d743e86b1cb06b11","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-02 00:51:38.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c494476cffbf041918d3957eb7f1062b112f6a85","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-08 13:53:23.000000000","message":"Patch Set 1: Code-Review-1\n\nShould include the releasenote tweak from\n    https://review.opendev.org/#/c/731875/\n\nwith this patch.","accounts_in_message":[],"_revision_number":1},{"id":"0da320446a6aa9adb1025b1bab280f9a47af848c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-07-07 20:12:12.000000000","message":"Patch Set 1:\n\nhttps://bugs.launchpad.net/cinder/+bug/1886222 needs analysis before landing this.","accounts_in_message":[],"_revision_number":1},{"id":"be9efccdbf6e6cb986cfa416eabe12f80861b9bd","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-07-08 07:27:44.000000000","message":"Patch Set 1:\n\n\u003e https://bugs.launchpad.net/cinder/+bug/1886222 needs analysis\n \u003e before landing this.\n\nIndeed that\u0027s a regression introduced by this patch.\nOnce it\u0027s fixed in master i will backport it to ussuri and then squash both these changes into one patch and propose in Train.","accounts_in_message":[],"_revision_number":1},{"id":"250c2449f1de0d4e03d9eb8f67cf7a5759bfc6c8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-19 16:44:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56ddcc57b59804421d67f109cbfb934b0a9cb59c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-19 16:47:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"183c3659bfba67d014ff7c80058ebf96c465d889","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-19 16:49:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/7a3ec7d : FAILURE in 4m 19s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/a8a5389 : FAILURE in 4m 26s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"4e72cfc25a337ff7eb1c55144fc591575c2e592b","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-19 16:50:51.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/595f1ebb9a21434c79044c91ef3bf12675d41359/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"b2895ed7019206303f67156d8e7381550e495c82","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-19 16:51:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71a32dc2ade8a4ddf3c6294b160b9a78ba8b8fce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-19 16:59:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c14c4ead265ee040d53d5b7338fc8de9f2d9dd05","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-19 17:27:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/00/732400/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d3dbd28/d3dbd28.html : FAILURE in 39m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e8a73343630110e180d61e11c2a2872f9192ae3d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-19 18:11:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/2/iscsi/ : SUCCESS in 1h 23m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/2/smb/ : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0a4304c185503bf35723e6e03be5d7cbb801d1bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 21:08:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bc0d15d0bb04c3bbf0a5a75c9edfec3 : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05af1cfc5fec4189b623d8635183ad57 : SUCCESS in 10m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/042647881fb84e518f4a3f6fc03cc81a : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91b85bc981bf4c01b1dfe375f719cd07 : SUCCESS in 12m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e722bc29ae6a4fccb9743dad910b8939 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90e57b4b6d34e83ba79714ee7dcc0c3 : SUCCESS in 16m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d908635c55354190afafedd84cd4e6e3 : SUCCESS in 1h 11m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b8504fb78a54110b2a7949e19476b59 : SUCCESS in 58m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/43981b01d3f04aa3844930da7cc97c67 : SUCCESS in 1h 05m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/161e5e02004d4c69a973d78d3b5d42e1 : SUCCESS in 1h 14m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00f79a4ad12d43f29a9d9a0f2db31190 : SUCCESS in 8m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b08231a233804416a85f65b1d95157f9 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3404779ea4874500b51c4d490bc63adc : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3f5c5033bca74a2795ee34c15dc510b4 : FAILURE in 22m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b568e14cdff947ccb27c3cd9a7ad4b42 : SUCCESS in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f5a4e7a1781e492b8dde45b699892bce : POST_FAILURE in 13m 11s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f372be7ab6b44ee6bca143fa18da9a90 : SUCCESS in 5m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a9f918311fbc419aac92c97ce58f2d48 : SUCCESS in 1h 33m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/411d082640ac4414a65ef32737fbefe9 : SUCCESS in 1h 16m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac056d266b4b4098a27dc06d69c1e42c : SUCCESS in 1h 20m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eba9516df89a42a793af25a0287247ea : FAILURE in 1h 41m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0b7faff3b58a4ec4a721a20c993766bc : FAILURE in 2h 28m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5ea78ddea8c9479bb0117fcf06edd689 : SUCCESS in 1h 13m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ca3aeda8f0341c484519b779b1c0140 : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":2},{"id":"55d219cf13e956f72d58e281da49316b6684ab0c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-19 23:28:27.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/2/check/tempest-dsvm-full-bionic/3ed59f2/ : TIMED_OUT in 2h 37m 42s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/2/check/devstack-plugin-ceph-dsvm-bionic/daba99e/ : SUCCESS in 1h 33m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f87cf9ebf0ba663560f2f3791713264def058140","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-19 23:31:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/732400/2/check/dsvm-tempest-cisco-zonemanager-job/debaea0/ : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6b9ecad76c75a471bc20102d6ea9735cc16e26ac","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-20 02:00:25.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0777a474d52f3e4ae4c3b3f331e6cc72bea25e8f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-20 08:03:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"808e8ea4a9fde9f0fb079d7fcb3e5fa4661dc83e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-20 08:17:43.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"bd57c3ff514323be7ca7951640e3ac9c93eda2ef","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-20 08:23:32.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f4913bf : FAILURE in 5m 00s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6472107 : FAILURE in 4m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"dbbfc2015844871dc061e00b27d215ec15ddc581","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-20 08:44:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/00/732400/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/209f4c4/209f4c4.html : FAILURE in 39m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ccdda1418fc1bcae2abdf7d5471c13139936d7d1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-20 09:10:22.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/3/iscsi/ : FAILURE in 1h 06m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/3/smb/ : FAILURE in 1h 05m 15s","accounts_in_message":[],"_revision_number":3},{"id":"56cf89606afbaf596396e497778ff09e1dc02bda","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-20 09:17:11.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/88adf5e3e5972c2f7a1405c23f1b932574349de8/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"d78fe635c95cded11b859550a69834f2a860842b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-20 10:14:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/732400/3/check/dsvm-tempest-cisco-zonemanager-job/7fe0619/ : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":3},{"id":"24d2dd6c86e462d7994b8313a9882f1e05c57234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 10:25:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b967ea2e36048a38b481e4d897b1271 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517324ef57e34adaa71b247afe377a78 : SUCCESS in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9efefc91e66d45a396256f06f191f83d : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c45a18aa1989455b8f7af2a915919aab : SUCCESS in 8m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8aa81e82defd4caea0d6483b03dd4f51 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d6f634d0984d45b17449e795e14c5e : SUCCESS in 10m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d97006752ec240d2aa3f9ea516e4c715 : SUCCESS in 1h 38m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3ab6fa9c79f94dd295e5cdfe510f4636 : SUCCESS in 37m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/1475b17e7f424a999c39c6e8a8c32474 : SUCCESS in 42m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f17dce347dc4d87a07c4e28f265f7be : SUCCESS in 1h 15m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45054780245e4159a6b9e07faa4cb462 : SUCCESS in 7m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/417e069119fe436ab06135e9756aac23 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0377786e1dc543dc9f6e9f9d2fde647c : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b99fd393ed2147d6bb552451a851755d : FAILURE in 13m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8941cda060014392ab7c06e84b55ddf2 : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/257d7e1d5c54457a9d0e39be7689e246 : SUCCESS in 1h 18m 38s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2d780a78a98d4f719967c892a86ed756 : SUCCESS in 5m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00ed6f2b745c46989f03fa2a80ea3804 : FAILURE in 1h 34m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2e817c29202e42e4b4221f8f17b99b61 : SUCCESS in 51m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1f20226b152414f99976fd65c8e7bf2 : SUCCESS in 1h 23m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a99641eeecc40598372b841baf4406d : SUCCESS in 1h 39m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/98dcd44b303d474c8f9d413a4fed16f1 : SUCCESS in 2h 14m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e9eb531d4b234c7999c4702e7d72e330 : SUCCESS in 39m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a8c5f27c2df4fe4bfccf509946b0547 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":3},{"id":"049917155e5387ab3c7b0a3c29b997dda46c7374","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-20 11:17:28.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/3/check/tempest-dsvm-full-bionic/8ac5a43/ : TIMED_OUT in 2h 39m 36s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/3/check/devstack-plugin-ceph-dsvm-bionic/750dbd2/ : SUCCESS in 1h 51m 15s","accounts_in_message":[],"_revision_number":3},{"id":"5750811ad6a8435974b0ed05e04ed8aad857c8ad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-20 11:24:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bc7d84550c0dd346ee0802fe7b78aa7c8c26b0bd","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-20 11:29:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/af142ef : FAILURE in 4m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/0ba0006 : FAILURE in 4m 18s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"ff84c88300aaeaeb22f1c9761cb4453e0dd4492f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-20 12:09:32.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"450e7a778425ffc7ee02237199fea10f7bd32510","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-20 12:44:24.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/3/iscsi/ : FAILURE in 1h 19m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/3/smb/ : FAILURE in 1h 05m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6a5a04902036216cf88156a46dfb9069c6dd09ce","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-20 12:55:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"385f7507d162cc5e6afe7998eacf872f49c60d43","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-20 12:57:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1515/index.html : SUCCESS in 1h 22m 44s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1659/index.html : SUCCESS in 1h 44m 54s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1677/index.html : FAILURE in 1h 26m 59s","accounts_in_message":[],"_revision_number":3},{"id":"6037df6c0b3f4dbefd0c1cd9d5ffba62c0b17b11","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-20 12:59:56.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c996355 : FAILURE in 4m 33s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6dd9922 : FAILURE in 4m 33s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"428d73940d98bf931ed2750bb5eb901f4e315509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 13:02:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/364bbdc138534bb296a9f74d2822c505 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beaa34f034a443ca8bc69479311a42f6 : SUCCESS in 7m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58f9ab8b2c8849629727f9530d8d0ac9 : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e412b00c57a14bfcb8a84ccf9ccfbefe : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8cdca45676747a5aff1ceaa4ebd29f7 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d9d4d28ccb24f1caa6990b6acb6e01d : SUCCESS in 13m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dcad76b2f64049f2b85dea3bad79bb69 : SUCCESS in 1h 17m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1bfc8a2dcc6b4c289c56429286c3f41b : SUCCESS in 48m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d9c34b8700284e75a2a6c1faa9ac1a16 : SUCCESS in 52m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6703538ad73d4d488da9a27cf0fbc3f2 : SUCCESS in 1h 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ac93ba4079842108bc0e6aafde14b63 : SUCCESS in 7m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a3a9f5cd76647d6b87b060ba3c22125 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6e557b7f14547b897b644de9ac734d5 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7b69f510812746f3aec4feafabe2abf3 : FAILURE in 16m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/831caba909a34ece8b9596bd23672452 : SUCCESS in 4m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/796a33c2cf304257b8cbfbcedcbfe327 : SUCCESS in 1h 28m 45s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c21e55675cc6408fa3cf2fe646b257c5 : SUCCESS in 5m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e2db73f754048e0ae599e4262fbd611 : SUCCESS in 1h 01m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fbbbb488e5f54ebcb46d4c9aa1dd2b70 : SUCCESS in 1h 09m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4be015f1bfb54925b9e0826f14c1f027 : FAILURE in 1h 20m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b41b3a00ee942f99dc3ee804c2adf81 : SUCCESS in 1h 31m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6dd1b05ceba433eb4c1b6f35ad1b45a : SUCCESS in 1h 26m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f209f4a22d3c464498f17384cbd3b504 : SUCCESS in 40m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aae65ccfaeba4a75adad6f6c077077b4 : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":3},{"id":"fc76387e5d689588e13a22247352041cb3e3ea72","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-20 13:08:12.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"96357ee137fd6c5597d09e433ce3c5b14fee1678","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-08-20 13:08:32.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/00/732400/3/check/synology-dsm-cinder-iscsi/c283a03 : FAILURE in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"4462649a2ae0a15b94c0283809abf1d9dd7d6265","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-20 13:09:18.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"dd8287687b34ac7d88c8efc6ca319bef79f51497","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-20 13:14:07.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/0ef576c : FAILURE in 4m 25s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/1daa81f : FAILURE in 4m 27s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"337f0f673e9f642dbceb7c9a7be17b312b978442","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-20 13:14:22.000000000","message":"Patch Set 4: Code-Review+2\n\nFix is good to backport.  Thanks for the references to the squashed patches in the commit message.","accounts_in_message":[],"_revision_number":4},{"id":"efa7f5e64944b7d1049030ca979017ee3b9d5bd2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-20 13:52:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/00/732400/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9baa160/9baa160.html : FAILURE in 39m 54s","accounts_in_message":[],"_revision_number":4},{"id":"ca52c0dfb83c4667a482df9772db527adaf1fa2c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-20 14:17:32.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/4/iscsi/ : FAILURE in 1h 01m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/4/smb/ : FAILURE in 1h 07m 44s","accounts_in_message":[],"_revision_number":4},{"id":"0916d819256254d4641b295ad3c3bd2ed13433a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 14:55:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbb34f8028ef48deae4d39149a0376e9 : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7025c2cc7745f9ab7d48551587ad3e : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91af9a14c521462eb4fc07761772bfd9 : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eca0279e0a14012aec3a2c196c76bf6 : SUCCESS in 9m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/245cdc1a47814f6cab40a3f5bf2efcaf : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a56418f2864dc4bddfb9940472e85b : SUCCESS in 16m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/26ec3c15e663411a9fa63c6348e7c829 : SUCCESS in 1h 30m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/df1b363ae7c042bebe457fa681930cea : SUCCESS in 41m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/3f03625f7213433bb09a9983fb8cfe8f : SUCCESS in 1h 07m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05c613b51f034f2a96e3920856a3f8a4 : SUCCESS in 1h 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/533439dc81db4ebcb83c6403c6bab079 : SUCCESS in 7m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89add5e4fc844ecca93eae4989de5c80 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15d652f2504e405fb34dc1597b02123b : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ec226319d8344aabe8e426d0d9d7e70 : FAILURE in 16m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/719215d46ffa4e108c7c87d609d15edb : SUCCESS in 7m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0e62808a55cc49ed8e03e27c4cd377d1 : SUCCESS in 1h 16m 09s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac5436ff75884d1c8bbd3075ace9020c : SUCCESS in 7m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6046cd4ba4cc44588b8214d6452c1be4 : FAILURE in 1h 35m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcdff6151eeb4ffb9c21e873cac55c1c : SUCCESS in 1h 13m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c174abd8ab0449d4b0d035f1e2d56481 : SUCCESS in 1h 04m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1329fae3527944ed976fa4a0f8b7a00f : SUCCESS in 1h 25m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/215e6b9ca74d4031b2b6aed8b1bdc4e4 : SUCCESS in 1h 25m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/80c0ede550a84ff28dcd70bedc575a68 : SUCCESS in 59m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/689801c610974efcb8b05de3656d8685 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":4},{"id":"4159f38879ba181833203d31d267111555092244","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-20 15:03:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"671adb868023869e190e14c8402318127799b00a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-20 15:08:42.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f467e87 : FAILURE in 4m 26s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ce81b1d : FAILURE in 4m 25s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"fbc91144d58274200e23e263bfab0bccfc843fef","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-20 16:31:45.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/4/iscsi/ : FAILURE in 1h 21m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/4/smb/ : FAILURE in 1h 27m 19s","accounts_in_message":[],"_revision_number":4},{"id":"2af85f3ef61321dc1968b508f5e777b7dc6086a3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-20 17:08:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/732400/4/check/dsvm-tempest-cisco-zonemanager-job/11dd39c/ : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e2e4b6ab1656e3ee4c1749f79aca559274a8262a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-20 17:21:24.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/4/check/tempest-dsvm-full-bionic/f760a3c/ : TIMED_OUT in 2h 38m 19s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/4/check/devstack-plugin-ceph-dsvm-bionic/5bc4fe0/ : SUCCESS in 1h 36m 23s","accounts_in_message":[],"_revision_number":4},{"id":"404bd796522d7422b93dd79c5abe0d1041323fbd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-20 17:51:42.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c433510ef1698c06b027d7f9ea0178684451b482","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-20 18:27:33.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e477d7f1df4644fe4f7be5514349c7a302ed976c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-20 18:39:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2c19309f3b22d16f627756ad52d02526bc852d3e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-20 18:44:20.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/00/732400/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/a2b7b08 : FAILURE in 4m 58s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/00/732400/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b092349 : FAILURE in 4m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"cef506a1a33e003340855b1ad086aa78be8fa090","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-20 19:55:24.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/732400/4/iscsi/ : FAILURE in 1h 11m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/732400/4/smb/ : FAILURE in 1h 15m 56s","accounts_in_message":[],"_revision_number":4},{"id":"f989e27f8c8361e7719367f57a88a596bcc08be8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-20 20:21:35.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3f2683fe5e67dc1ac46da7e62d71de43a893ca90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:59:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e99f503432744b31983d5330cd274a16 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b63f10507d1342e8affeb5ab0458ea12 : SUCCESS in 10m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/309d0341888442be9bb37a70d1858059 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e91fd2034124a3c9bb6c407bf8d8ea5 : SUCCESS in 9m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdde7ab7f2c543c8b8698a9ae8cfe0ed : SUCCESS in 12m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e71b4e9b644a579058e4a2c46974b1 : SUCCESS in 13m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f56d6bb513da4d6495feace7ed4ff2dc : SUCCESS in 1h 40m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/84271821db1243bba46feff267acdbbf : SUCCESS in 56m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/035ac45bae18471d82e559a392b9438d : SUCCESS in 1h 04m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b8811a9ce8c4d1ca59e296c8cd1fc7c : SUCCESS in 1h 18m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9c878a8a97a4252a039ef4f24b87951 : SUCCESS in 9m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/46cfb4dc179d433ba34949040a0b6e90 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76a0f2fcfbf64778a6ff9ffa4545c254 : SUCCESS in 7m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9f53c6c8a0bd44f8a1ebf31a28994af9 : FAILURE in 20m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b431dae7e4f34ea5b7256e6e0f947497 : SUCCESS in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea26dde82cd4403980324643997ffe02 : SUCCESS in 1h 15m 13s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6cbb6f6086d44af5b8ed7a09a3b54cef : SUCCESS in 8m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/435ba798e172433f965e77a77bb7aa28 : SUCCESS in 1h 27m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ddb44298c1d04c9aa4fb26600f0c14ca : SUCCESS in 1h 10m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1866468ff8ba40a998fea9d8376285a7 : SUCCESS in 1h 23m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6f2ef75af7f644428c6f4bebe044bcef : SUCCESS in 1h 36m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c20ba66bd274f14ad8d4b0db46e31da : SUCCESS in 2h 03m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5e579ac1d905447da4561f6e90a46e53 : SUCCESS in 1h 03m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/308f48fa9e574d29a03b2555afcb9a15 : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":4},{"id":"5aee3770a57e23d238b5e7595375939bb0b44c74","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-08-20 21:03:19.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/00/732400/4/check/synology-dsm-cinder-iscsi/4ff040e : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"9937b4cb624cc3a31c96054302bbd52a0de031f7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-20 22:56:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/00/732400/4/upstream-solidfire-check/cinder-solidfire-iSCSI/dc71e82/ : FAILURE in 1h 56m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"54b0430efb49bd65fc0eeee5f65b1912cba99c94","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-20 23:28:51.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/4/check/tempest-dsvm-full-bionic/79927b4/ : TIMED_OUT in 2h 39m 55s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/732400/4/check/devstack-plugin-ceph-dsvm-bionic/0e3addb/ : SUCCESS in 1h 37m 53s","accounts_in_message":[],"_revision_number":4},{"id":"53d567b7a8b7b4a679820bf8807cfd0cfc934a6d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-20 23:47:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/00/732400/4/upstream-check/cinder-cDOT-iSCSI/eb8beec/ : SUCCESS in 2h 46m 56s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/00/732400/4/upstream-check/cinder-cDOT-iSCSI-IPv6/a81cdc5/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/00/732400/4/upstream-check/cinder-cDOT-NFS/56ca3f3/ : SUCCESS in 37m 24s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/00/732400/4/upstream-check/cinder-cDOT-NFS-IPv6/d4bfe0a/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/00/732400/4/upstream-check/cinder-cDOT-FCP/d3ac8b2/ : FAILURE in 1h 36m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"d9912d9f01607760c8fb1c5b7b9911ae0791d495","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-21 00:28:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/732400/4/check/dsvm-tempest-cisco-zonemanager-job/2831e32/ : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":4},{"id":"ec865c4b19b18dd6454bfc9809336beea93ce332","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-21 02:21:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1521/index.html : SUCCESS in 1h 24m 21s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1665/index.html : SUCCESS in 1h 49m 27s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1683/index.html : FAILURE in 1h 52m 44s","accounts_in_message":[],"_revision_number":4},{"id":"fd3ea6c3d4331c2fb1560fbd1e537deb28273013","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-21 08:08:39.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/00/732400/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"4ee2b6bc943ad64241e32a57b5c7fb9b6ad36201","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-21 09:56:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/00/732400/4/check/dsvm-tempest-zadara-driver/258bacf : SUCCESS in 1h 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c92b2def483694da85f53258ff8d136c858d60a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-08-21 13:00:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"759b60d6743f8aa77f183d018b96bde0ce4a3b57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 13:01:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f21ab807dade87f387bc78e0a1c1efc41fc41887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:40:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/318dd4d722164970bf4beaceaa20e43d : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfdabffc44ea41b6ba903033d1766fbe : SUCCESS in 9m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee75566681674146886ebb5f28393165 : SUCCESS in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78b9580a690a4664ac4cc4ccab1d8bc9 : SUCCESS in 11m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8214561456914dbf8843f9bcfb48ea27 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8514f93ee4048558d80715073805118 : SUCCESS in 11m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/28fab7c6f2a14d4fa2b21b18c69bfb07 : SUCCESS in 1h 31m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4fbc389bfca4d68ab0754f8782c66dd : SUCCESS in 56m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/4fd9220d972840d889dab8704bd7e502 : SUCCESS in 1h 14m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4101311458e14531a8165e5daabcc934 : SUCCESS in 1h 23m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9b9567e378245ba84f66b3eee5539be : SUCCESS in 7m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3fb64c2078454968b6c0b343580b4e53 : SUCCESS in 1h 22m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6bfc93d00f85456ea3e7c16aa8e8a6fb : SUCCESS in 1h 02m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97cd5a8e2cf949e4ae15688706301e17 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":4},{"id":"3b197df6699661b76c9b365c62ba8dfac5a4ce1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:40:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0297e8ef2c023427db6fba091e8dc87a7817ac63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:42:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d8ee13fab8432a8edabc7ec716b1aa : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dc0ed96984243bea443a9cb8624f72f : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4},{"id":"f3aa81119b20cc2cf13ad5608ddac06d13c703b5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-21 15:04:54.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 14f5a20dadb6590941633035a1dd8bcb0a7cb3b1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"193f3e97ce19bd8926bac23781e51d68de8ddcd4","revisions":{"075fe1ff80ed9509d4e8ddd41f8670c72d5567d1":{"kind":"REWORK","_number":1,"created":"2020-06-01 14:59:39.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/732400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/732400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/732400/1"}}},"commit":{"parents":[{"commit":"adecc819869e4cf0b3c687e7ce4ea1340d6e7654","subject":"3PAR - Fix renaming volume after migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/adecc819869e4cf0b3c687e7ce4ea1340d6e7654"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 20:24:48.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-01 14:59:39.000000000","tz":0},"subject":"Creating image-volume cache on NFS backend fails","message":"Creating image-volume cache on NFS backend fails\n\nsnapshot.id doesn\u0027t have `tmp-snap` in it as id\nis not passed as a option while creating snapshot\nobject. This should be change to\n`snapshot.display_name` at same location\n\nCloses-Bug: #1875570\nChange-Id: I66da0212a61d778fe9795d382b3af5b225a8aa54\n(cherry picked from commit e52ab1a3adbc8462b02c342ee7990e527a9f73bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/075fe1ff80ed9509d4e8ddd41f8670c72d5567d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/075fe1ff80ed9509d4e8ddd41f8670c72d5567d1"}]},"branch":"refs/heads/stable/train"},"595f1ebb9a21434c79044c91ef3bf12675d41359":{"kind":"REWORK","_number":2,"created":"2020-08-19 16:44:50.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/732400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/732400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/732400/2"}}},"commit":{"parents":[{"commit":"0184070e19910c8583f7d2e540959066160a09ef","subject":"Merge \"PowerMax Driver - PowerMax Pools Fix\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0184070e19910c8583f7d2e540959066160a09ef"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 20:24:48.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-19 16:41:16.000000000","tz":0},"subject":"Creating image-volume cache on NFS backend fails","message":"Creating image-volume cache on NFS backend fails\n\nsnapshot.id doesn\u0027t have `tmp-snap` in it as id\nis not passed as a option while creating snapshot\nobject. This should be change to\n`snapshot.display_name` at same location\nAlso fixed an issue with creating a snapshot on an\nNFS backend if the snapshot name is not specified.\n(I33da9c65ef56b3e4967170e3a0fb25f12e067876)\n\nCloses-Bug: #1875570\nCloses-Bug: 1886222\n\nChange-Id: I66da0212a61d778fe9795d382b3af5b225a8aa54\n(cherry picked from commit e52ab1a3adbc8462b02c342ee7990e527a9f73bb)\n(cherry picked from commit 11b5c9d97bd69f9e8b5db0d32bfe257063170155)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/595f1ebb9a21434c79044c91ef3bf12675d41359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/595f1ebb9a21434c79044c91ef3bf12675d41359"}]},"branch":"refs/heads/stable/train"},"88adf5e3e5972c2f7a1405c23f1b932574349de8":{"kind":"REWORK","_number":3,"created":"2020-08-20 08:03:21.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/732400/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/732400/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/732400/3"}}},"commit":{"parents":[{"commit":"0184070e19910c8583f7d2e540959066160a09ef","subject":"Merge \"PowerMax Driver - PowerMax Pools Fix\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0184070e19910c8583f7d2e540959066160a09ef"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 20:24:48.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-20 07:59:43.000000000","tz":0},"subject":"Creating image-volume cache on NFS backend fails","message":"Creating image-volume cache on NFS backend fails\n\nsnapshot.id doesn\u0027t have `tmp-snap` in it as id\nis not passed as a option while creating snapshot\nobject. This should be change to\n`snapshot.display_name` at same location\n\nCloses-Bug: #1875570\nChange-Id: I66da0212a61d778fe9795d382b3af5b225a8aa54\n(cherry picked from commit e52ab1a3adbc8462b02c342ee7990e527a9f73bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88adf5e3e5972c2f7a1405c23f1b932574349de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88adf5e3e5972c2f7a1405c23f1b932574349de8"}]},"branch":"refs/heads/stable/train"},"193f3e97ce19bd8926bac23781e51d68de8ddcd4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-08-20 13:09:18.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/732400/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/732400/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/732400/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/732400/4"}}},"commit":{"parents":[{"commit":"0184070e19910c8583f7d2e540959066160a09ef","subject":"Merge \"PowerMax Driver - PowerMax Pools Fix\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0184070e19910c8583f7d2e540959066160a09ef"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 20:24:48.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-20 13:08:08.000000000","tz":0},"subject":"Creating image-volume cache on NFS backend fails","message":"Creating image-volume cache on NFS backend fails\n\nsnapshot.id doesn\u0027t have `tmp-snap` in it as id\nis not passed as a option while creating snapshot\nobject. This should be change to\n`snapshot.display_name` at same location\nAlso includes regression and NITs from other\npatches\n(I33da9c65ef56b3e4967170e3a0fb25f12e067876)\n(Ia620043904f4f31a81bc8463fb36476a149f9808)\n(Id0b4981be5bf4e79aedcdad2258fc64aad8eeb72)\n\nCloses-Bug: #1875570\nChange-Id: I66da0212a61d778fe9795d382b3af5b225a8aa54\n(cherry picked from commit f68a64540f20967f933a3c9546837ddde23c20cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/193f3e97ce19bd8926bac23781e51d68de8ddcd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/193f3e97ce19bd8926bac23781e51d68de8ddcd4"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
