)]}'
{"id":"openstack%2Fcinder~733105","triplet_id":"openstack%2Fcinder~master~Ia7cc17472677a693c6162f0b8b0529df62eed7cf","project":"openstack/cinder","branch":"master","topic":"bug/1823200","hashtags":[],"change_id":"Ia7cc17472677a693c6162f0b8b0529df62eed7cf","subject":"Remove VxFlex OS credentials from connection_properties","status":"MERGED","created":"2020-06-03 11:14:03.000000000","updated":"2020-06-08 06:19:56.000000000","submitted":"2020-06-04 20:24:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733105-1591302298766-ce7c6379","meta_rev_id":"a79dd2cde2318cfff4c1bbb4381646c30dd247cb","_number":733105,"virtual_id_number":733105,"owner":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-04 14:22:21.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 19:49:15.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2020-06-05 01:08:18.000000000","post_submit":true,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-06-05 10:18:53.000000000","post_submit":true,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2020-06-05 13:08:12.000000000","post_submit":true,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 16:31:49.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 14:47:06.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-06-04 15:03:13.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 20:24:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-05 09:07:34.000000000","post_submit":true,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-06 02:37:34.000000000","post_submit":true,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-06-04 16:30:55.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 18:47:15.000000000","_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"date":"2020-06-04 15:21:50.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"date":"2020-06-04 15:09:50.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 20:00:37.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 19:12:47.000000000","_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"date":"2020-06-04 19:58:54.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 15:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-04 18:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-04 14:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-04 18:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-04 14:44:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 11:14:31.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-06-03 12:30:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-06-03 13:11:38.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-03 15:32:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-03 16:06:59.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-06-03 19:35:14.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 08:20:50.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-06-04 10:09:45.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 13:16:08.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-04 14:20:14.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-06-04 14:22:21.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 14:47:06.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 15:03:13.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 15:09:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-06-04 15:21:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 15:48:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 16:30:55.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 16:31:49.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 18:47:15.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 18:58:24.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-04 19:12:47.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 19:49:15.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-06-04 19:58:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-06-04 20:00:37.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 20:24:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-05 01:08:18.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-05 09:07:34.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-05 10:18:53.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-06-05 13:08:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-06 02:37:34.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"}],"messages":[{"id":"509b21591c6402064cc485156a61117f0637e5f1","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-03 11:14:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b92492152610959cd74b91207fb8d04a7c876df6","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 11:25:32.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/5050902d482b8404e6c3375bd51b2d3364db5d3c/index.html : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2c5bc9ef04893eacf485ff3d494e04d95269158a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-03 11:27:13.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"80e6b9b188d030a588057319bfb47052254c64cd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 11:30:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c49085822dd1211aa2662a2eeed3a990f01248c2","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-06-03 11:34:02.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"cbd5039b9b0fb958269c9e212084bfdcb01ebe72","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-06-03 11:50:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/05/733105/1/check/fujitsu-eternus-dx-iscsi/34fa454 : FAILURE in 36m 00s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/05/733105/1/check/fujitsu-eternus-dx-fc/cfcea35 : FAILURE in 27m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e0bd216b752f90e21373f43d20b67dcb6083bbee","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-03 12:06:02.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/733105/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"18af40c23572e2cd3baacbc9480438bbd8d29fb9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-03 12:17:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"6b7af9ab46d20f1fe1cd10b3d05a439f39a7c43a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 12:20:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96ba27647760a8ccdbdbca97f30f2ce6c11f6c36","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-03 12:20:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/733105/1/check/dsvm-tempest-cisco-zonemanager-job/6bf66f1/ : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"da5ea065dd2b5dbf47c490715411d7ab293b3e5a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 12:22:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0066d67456f08706faf2b71670b96a097645f79","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-03 12:24:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a811ec6a212349b12de39c19780542ef63e11d8","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-03 12:25:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a59ab34d5d70e8770dd94fd0d977a55f6e771948","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-03 12:25:59.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"eb939d363ff975cb01cf996ffcb7cc271d6f2be3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 12:26:30.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks Ivan!","accounts_in_message":[],"_revision_number":2},{"id":"4df1ea4ef9d3b92291f1f2dac5f056445aa68195","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-03 12:30:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ccfed253575bf9b110fae4981f1e514872b42f2a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 12:52:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9e4b663d0ece2fe7ba14d022c6e96c40f252fed1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 12:57:57.000000000","message":"Removed Workflow+1 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"d738ce991a095cb7aeb593f3a7e1c80a6d26a677","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-06-03 13:01:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/05/733105/2/check/fujitsu-eternus-dx-iscsi/a4065b3 : FAILURE in 35m 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/05/733105/2/check/fujitsu-eternus-dx-fc/ddb6d15 : FAILURE in 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d20476cf79634e632f2ee96008ac15664f243475","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2020-06-03 13:11:38.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- Cinder-AS13000 http://218.29.111.78:9000/05/733105/2/check/Cinder-AS13000/1be18ef : FAILURE in 13m 02s\n- cinder-test-fc http://218.29.111.78:9000/05/733105/2/check/cinder-test-fc/a9693c7 : FAILURE in 5m 34s\n- cinder-test-iscsi http://218.29.111.78:9000/05/733105/2/check/cinder-test-iscsi/9dd0383 : FAILURE in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"339a6c23eeaa54af4700ce01ef039ec5137b5c19","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-03 13:17:12.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/733105/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"6da889e21845123ddd549422d231ff756dcdee42","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 13:28:59.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/52be8aff00a3debe90811d6e9cad7032f075b026/index.html : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ecfcc3fdfc18bbd30cf13606d3e9e0c2d9cb87f3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-03 13:35:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/733105/2/check/dsvm-tempest-cisco-zonemanager-job/99bcbd5/ : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":2},{"id":"14547c60761bc870f6d3c5bf9ca6c85578d6d5d6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-03 13:35:28.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-733105-1/ns5_nfs : SUCCESS in 2h 1m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-733105-1/ns5_iscsi : SUCCESS in 2h 21m 4s","accounts_in_message":[],"_revision_number":1},{"id":"efdfd17c49f3e4d26b4ef6c6047be16479a26da2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-03 13:36:20.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/733105/2/iscsi/ : SUCCESS in 1h 10m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/733105/2/smb/ : FAILURE in 27m 07s","accounts_in_message":[],"_revision_number":2},{"id":"fae33a0e44d18f4a8d259f1b2d6dc9fe99095d8e","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-06-03 14:07:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/05/733105/2/check/nimble-iscsi-driver-dsvm-volume/81f4b1f : SUCCESS in 1h 41m 57s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/05/733105/2/check/nimble-fc-driver-dsvm-volume/3fea449 : SUCCESS in 1h 33m 21s","accounts_in_message":[],"_revision_number":2},{"id":"16af8a0b41e03bc80652fdb674690923d1859de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 14:28:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59ff9e67674e48969e036e5609779b56 : FAILURE in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dccbb100c964f1d8517a57ffef40187 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e93b8cf777d14cada9d950eff471f9dd : SUCCESS in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a38f5d3ea17c4a6f9dd6f7d5b8fbf418 : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fcd63fd47404bc7a285b2b394cfa530 : SUCCESS in 13m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/232374c26e154bbc94d19db276ebbc40 : SUCCESS in 1h 05m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36507290a56a4e2291e6ad39d575c715 : SUCCESS in 1h 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9a9564bad844ba7a7828d0bc17d317c : SUCCESS in 8m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/57b055f67e03463387aa88f5c0f5f10f : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99df733321654e0aa183af47c4e0e4eb : SUCCESS in 6m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fbc01eb3eed549c3bd9b30193e3db418 : SUCCESS in 7m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cdd03fbf975245ca81bbe55a36b930e9 : SUCCESS in 39m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71f3413a738d4186a9e582cd52655df2 : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e44431ff871a4eedbb92e02febce7a94 : SUCCESS in 1h 20m 43s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db0c15dd7de14ed29843a3b1c88b72cf : SUCCESS in 1h 20m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2e60d8e4b866429f9bcac2105e93170a : SUCCESS in 1h 07m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/79d1da14a6474402a86e7e1c71e2d309 : FAILURE in 1h 19m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c95c3e7a860745c28fd53757d8112999 : SUCCESS in 1h 12m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c5b818635314d50bacacc62087a0e13 : SUCCESS in 2h 01m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8f9e99d151c49b38d0488e254df1dd9 : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4638a71d8c39302aeab2a6c2666eac2073024141","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 14:46:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/733105/2/check/DellEMC_XtremIO_FC/e6ceaf5/DellEMC_XtremIO_FC/112 : SUCCESS in 1h 31m 35s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/733105/2/check/DellEMC_XtremIO_iSCSI/fa60e45/DellEMC_XtremIO_iSCSI/111 : FAILURE in 1h 56m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"54b563c2d365c42b39f363e28f61a260eeaa3822","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-03 14:51:48.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-733105-2/ns5_nfs : SUCCESS in 2h 26m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-733105-2/ns5_iscsi : SUCCESS in 2h 2m 26s","accounts_in_message":[],"_revision_number":2},{"id":"dec16a61da76c6ddbccab3bc68ecd3f0e0f14f11","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-03 14:52:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9c97d83a69f73311b09e848f64c1a291e781da18","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-03 15:01:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e1fd7ba5cb96509bc7a72b978ea2d83b0e1548f5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 15:32:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/733105/2/20200603215023/NEC-ISCSI : SUCCESS in 2h 32m 50s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/733105/2/20200603215023/NEC-FC : SUCCESS in 2h 42m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a65a4e3bbcda9de45b4577b46cef8393f6aa3981","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-03 15:40:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/733105/2/check/cinder-storpool-tempest/163f611/ : SUCCESS in 2h 36m 53s","accounts_in_message":[],"_revision_number":2},{"id":"15040ddec6623e6d00a722a59064e9e809e589d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-03 15:48:03.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/733105/2/check/DellEMC_Unity_FC/a6dfd78/DellEMC_Unity_FC/102 : SUCCESS in 1h 26m 28s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/733105/2/check/DellEMC_Unity_iSCSI/50dd19d/DellEMC_Unity_iSCSI/97 : SUCCESS in 1h 34m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f5c64efdda9094d14a1fcb2ce6ef84305922f65d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-03 15:50:58.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/733105/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4034b520f6fd27c0fa3502c8f93b680d128ab65a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-03 16:02:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/733105/2/iscsi/ : FAILURE in 1h 00m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/733105/2/smb/ : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":2},{"id":"003237d028ed1de8c0f5c4e8fd4b51e446ecbb7b","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-03 16:44:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1122/index.html : SUCCESS in 1h 21m 35s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1273/index.html : SUCCESS in 1h 42m 39s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1299/index.html : SUCCESS in 1h 30m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f13124ae5292bf746db20394e0cb4ec5ac554202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:10:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0445471ddcf04eb79b886a319bff1c38 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/583e776976cb46b1b09293ae07c764eb : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a0b979801974c6687d767361baf957c : SUCCESS in 12m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85ac02d61a5a44669074491cc7eeeb7a : SUCCESS in 14m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd0a591414640648bed8055da3d1752 : SUCCESS in 18m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/6f6ec78d422a4a4487af846ecdfa4779 : SUCCESS in 1h 00m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de0d9c55298d40e7bec85eef2abeba92 : SUCCESS in 1h 19m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c1595e743ca4a1ea2ed5bca55b82dcb : SUCCESS in 8m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/82f8cca9e19e4a22a6cb9937d66d0e5a : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08b48683fe3d4bb2bb2fe8e99e5091ad : SUCCESS in 13m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ebf3e5fae494a0faf3ff6f2bda59152 : SUCCESS in 7m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1cfa3a62139246c9a8ee29b0922da0a4 : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/faf46c58f4334188b9ea4ab2e46c87c1 : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bb05ab11f1494b1a97f125d0025533be : SUCCESS in 1h 06m 48s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1f873f32b95f40668ab3d072e4e6d187 : SUCCESS in 1h 26m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fa146ee2e9fa4196b305ceadb766cd74 : SUCCESS in 51m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/65ed1b1a12534bae84f788b9243701d8 : SUCCESS in 1h 19m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e01c242116bb4418985c20d56207e214 : SUCCESS in 1h 15m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3a5c8ce7d72a41bbab18d481a326b959 : SUCCESS in 2h 03m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06f74e82ded44c28af81fa9d586d3f74 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ad02dc6fa7feb0ade05d722f6d8ca7d5ef86d106","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-03 18:14:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/733105/2/upstream-solidfire-check/cinder-solidfire-iSCSI/a9d83ec/ : SUCCESS in 1h 02m 53s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"13728c34c7340a171cf509b26362a409ac542d8a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-03 18:17:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/733105/2/check/dsvm-tempest-cisco-zonemanager-job/8012c23/ : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0cdb80b881e1240e6a8a7bb5dd737c9e0abc0281","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-03 18:18:49.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-733105-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"9e7d44e38d62df828cc9fdf0bce5f85f9b8b8d35","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-03 18:24:30.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/733105/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bcaae387e812353246ca6650b5bd468bbe0bafbf","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 19:35:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/05/733105/2/check/DellEMC_VNX_iSCSI/819acd2/DellEMC_VNX_iSCSI/74 : SUCCESS in 2h 24m 40s\n- DellEMC_VNX_FC http://publiclogs.emc.com/05/733105/2/check/DellEMC_VNX_FC/ed08cc6/DellEMC_VNX_FC/75 : SUCCESS in 1h 01m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"f0cd4b99fee27c88d473d42b332e4d8aa5a7e820","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-03 19:57:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/05/733105/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4177c3b : SUCCESS in 53m 57s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/05/733105/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/e3d68e7 : FAILURE in 1h 53m 20s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"3a8dff3241aad27dc95bae06fbf660aa64b873e2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-03 20:47:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/733105/2/check/DellEMC_Unity_FC/bc0be91/DellEMC_Unity_FC/105 : SUCCESS in 1h 22m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/733105/2/check/DellEMC_Unity_iSCSI/1540454/DellEMC_Unity_iSCSI/100 : SUCCESS in 1h 38m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4ee884b5bf361a8f2ff5913e904215fc656ca34c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-03 20:51:14.000000000","message":"Patch Set 2: Review-Priority+1\n\nThis has to wait for a new os-brick release for victoria, which will be 3.1.1.  (We had to revert the switch from retrying to tenacity in os-brick master, which is causing some test timeouts.  The revert patch is https://review.opendev.org/#/c/733396/ )","accounts_in_message":[],"_revision_number":2},{"id":"7eaccf0f4826867dac7a3d0ca5867d6b562bb8d0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 21:06:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/733105/2/check/DellEMC_XtremIO_FC/3993e46/DellEMC_XtremIO_FC/115 : SUCCESS in 1h 32m 33s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/733105/2/check/DellEMC_XtremIO_iSCSI/b643697/DellEMC_XtremIO_iSCSI/114 : FAILURE in 1h 58m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"7aef5de66b64965bfd2c6cc5f4270e47ec1d83e1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-03 21:07:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/733105/2/upstream-check/cinder-cDOT-iSCSI/2929c01/ : SUCCESS in 47m 01s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/05/733105/2/upstream-check/cinder-cDOT-iSCSI-IPv6/357f9dd/ : SUCCESS in 3h 29m 15s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/733105/2/upstream-check/cinder-cDOT-NFS/be10a51/ : SUCCESS in 47m 27s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/05/733105/2/upstream-check/cinder-cDOT-NFS-IPv6/a48f487/ : FAILURE in 3h 55m 54s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/733105/2/upstream-check/cinder-cDOT-FCP/d8f668f/ : SUCCESS in 1h 43m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"afce512087380bade0e9c22a9ca94d12e65d1894","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 21:11:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/05/733105/2/check/DellEMC_VxFlexOS/334bff4/DellEMC_VxFlexOS/83 : ABORTED in 4h 00m 23s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"17ea89696137cf5eb9b4c6be24d4a1b4ff9edba6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-03 21:57:35.000000000","message":"Patch Set 2:\n\nOK, I spoke too soon.  After https://review.opendev.org/#/c/733412/ merges to cinder/master, the upper-constraints patch should pass, and we should be able to use os-brick 3.1.0 after all.  So what\u0027s missing here is (a) rebase on master after 733412 merges, and then adding requirements.txt and lower-constraints.txt to this patch, changing the os-brick version specified in each to 3.1.0.","accounts_in_message":[],"_revision_number":2},{"id":"dbd61bb0e20a0ed14a219c41235d4224c867aad8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-04 02:27:17.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/733105/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"176831e7171a672c9402a8e91b0fe59d73085ad9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-04 04:24:42.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/733105/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f46ba7d1cd78c64a27ed4c6ad751a13be4b34c07","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-06-04 05:09:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7d3ef2e60f865e21343c85615595caa682a0cee7","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-04 05:23:49.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/39538f2d1f48f867a795172a8e1de9673d6cb1d3/index.html : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"177fa11c4550a4ea5a05fc416791bc54fa83c743","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-04 06:01:24.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/733105/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9cf0a75f99defed669db43c0b504fe22417f0b8d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-04 06:06:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/733105/3/iscsi/ : SUCCESS in 56m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/733105/3/smb/ : SUCCESS in 54m 53s","accounts_in_message":[],"_revision_number":3},{"id":"cb8ecf62cc3892572e02c07c9a3f10e13ff18d72","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-04 06:19:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/733105/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0aef444/0aef444.html : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":3},{"id":"25ec18d82841c7309208f5a2dd28c933de3973a2","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-06-04 06:49:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/05/733105/3/check/nimble-iscsi-driver-dsvm-volume/df506d9 : SUCCESS in 1h 39m 01s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/05/733105/3/check/nimble-fc-driver-dsvm-volume/c441d60 : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b1f7a7f68de275fa34fb1317dd0a50c6dcaf9427","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-04 06:51:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/733105/3/check/DellEMC_Unity_FC/0f59fcd/DellEMC_Unity_FC/110 : SUCCESS in 1h 27m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/733105/3/check/DellEMC_Unity_iSCSI/82b7da5/DellEMC_Unity_iSCSI/105 : SUCCESS in 1h 41m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a9e78493ad076b037be269a1f41f424cd95a9d21","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-04 07:09:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/733105/3/check/DellEMC_XtremIO_FC/8a522d3/DellEMC_XtremIO_FC/120 : SUCCESS in 1h 30m 14s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/733105/3/check/DellEMC_XtremIO_iSCSI/db11d22/DellEMC_XtremIO_iSCSI/119 : FAILURE in 1h 58m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"32f827b1bf69087908a05e8e7828f876b4002962","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-04 07:14:22.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-733105-3/ns5_nfs : SUCCESS in 2h 1m 51s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-733105-3/ns5_iscsi : SUCCESS in 2h 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"92643b7d223cd4f08ed4a298df4c5927d347253c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 07:15:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d06b9b261b474b448aae7d7e142ece1e : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3408abc0f75408ab90071cff7d2aac2 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45a989e0d12848b3a7949147cad8c6ad : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d308ec83ec74487982128de06a02f80 : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a06880ac07a49909d75976e999d58e7 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/ce48be0f59194987a1a0679c9061f78b : SUCCESS in 1h 05m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68c9113153f74f569b8d7e044df34018 : SUCCESS in 1h 15m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0975d89f6f324215b1f53f6f730972e5 : SUCCESS in 7m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/52434a94fe8c485590b53c22d8f621f1 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24380b90927247aa91ae9bfc62a0b021 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/caee0764beae47bf86d0f3b16f5db018 : SUCCESS in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/01ee576e4c03458c853edb07e7a72630 : SUCCESS in 42m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe0fcb303e964778884dbccef3f49ce6 : SUCCESS in 4m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/772d647a6e664ae98519056d30bf7bf8 : SUCCESS in 1h 09m 25s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/061a15504f874962bf584e4056d676e7 : SUCCESS in 1h 23m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4945081f8b841789de8714ca8a8da69 : SUCCESS in 1h 12m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/73321e1bf7924b2ca0a68dfde53ba1bb : SUCCESS in 1h 30m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ce4c3c70a0244dad92d938b11e68e60f : SUCCESS in 1h 23m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3f6f78351584b0588d3920ba98689c5 : SUCCESS in 2h 00m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fd4d16c17ac49a9b0a3e7a3a3669d15 : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":3},{"id":"e54bbaeb06a3f2c5870df52843c3fc08e6e74203","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-04 07:16:07.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/733105/3/check/tempest-dsvm-full-bionic/ff59158/ : SUCCESS in 1h 21m 21s\n- devstack-plugin-ceph-dsvm-bionic devstack-plugin-ceph-dsvm-bionic : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"be276210f05723ae32df60679dc7cae81e0d34a5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-04 07:27:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/733105/3/check/dsvm-tempest-cisco-zonemanager-job/fc242cc/ : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":3},{"id":"2a3db63b128e271914f5674af58c39400e9f194b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-04 08:20:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/733105/3/upstream-solidfire-check/cinder-solidfire-iSCSI/f6ccb53/ : SUCCESS in 58m 23s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"37f543c22054f0d7d61401101d1821ddb462c248","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-04 08:36:16.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-733105-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"d6de7b42f957b91633a3f078c3c245de26e87edf","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-06-04 09:34:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/05/733105/3/check/fujitsu-eternus-dx-iscsi/439d0ef : FAILURE in 4h 19m 45s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/05/733105/3/check/fujitsu-eternus-dx-fc/f80374b : FAILURE in 4h 19m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e6ff75a9f400f04fd850306c27b83dab3f4f174f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-04 10:09:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/733105/3/upstream-check/cinder-cDOT-iSCSI/f763d52/ : FAILURE in 33m 52s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/05/733105/3/upstream-check/cinder-cDOT-iSCSI-IPv6/fabe6d0/ : FAILURE in 6m 37s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/733105/3/upstream-check/cinder-cDOT-NFS/401f569/ : SUCCESS in 43m 14s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/05/733105/3/upstream-check/cinder-cDOT-NFS-IPv6/7738948/ : SUCCESS in 1h 25m 34s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/733105/3/upstream-check/cinder-cDOT-FCP/a399951/ : FAILURE in 2h 53m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b655bc75390d9d945d4f0a593a1b80b41969f0b8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-04 10:22:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/733105/3/check/DellEMC_Unity_FC/614514f/DellEMC_Unity_FC/114 : FAILURE in 48m 46s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/733105/3/check/DellEMC_Unity_iSCSI/fc620a7/DellEMC_Unity_iSCSI/108 : FAILURE in 5m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"1b2a6d4bbb3c31113f2f1e3b0b4aacfdc4fdf00a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-04 11:11:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/733105/3/check/cinder-storpool-tempest/ed4e186/ : SUCCESS in 2h 39m 29s","accounts_in_message":[],"_revision_number":3},{"id":"cdacee8596fe8728c488bb27dccf830df904bdd0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-04 12:13:15.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/733105/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"bec0661012bccf20f3056a785f62791b379d4a16","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-04 12:54:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/05/733105/3/check/DellEMC_VxFlexOS/892a749/DellEMC_VxFlexOS/89 : ABORTED in 4h 00m 31s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"8742c7516eb6f2cd0b86035b224f774fb6896587","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-04 13:07:18.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/733105/3/check/DellEMC_XtremIO_FC/77793ab/DellEMC_XtremIO_FC/123 : FAILURE in 40m 14s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/733105/3/check/DellEMC_XtremIO_iSCSI/4244892/DellEMC_XtremIO_iSCSI/122 : FAILURE in 1h 59m 05s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"ec27a7809c778aba228a048ee0a2e8061c1ecd8d","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-04 13:16:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"648ca716ccaad13253c07d7104c3d2ec183d884b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:16:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"43b7ca361c5bf55ac1335b5f6a7bb7a487ff3da8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:16:53.000000000","message":"Patch Set 3: Workflow-1\n\nI think we actually want to get the requirements update of os-brick as part of this first.","accounts_in_message":[],"_revision_number":3},{"id":"137a863dde7c2a44b26febe5b52d50f0067dea5d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 14:11:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"18ea3e7ee4ed9f07c51acecd026663d675a7b928","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-04 14:20:14.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3a0c40d477c27c72f64502ddf8be40b6ff7ac70d","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-04 14:22:21.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/b2c9592281b24824c2158f91a9295288f41dd81a/index.html : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c22d54b54ea82928de007d2698d68a56b383a9f2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-04 14:44:52.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\nThanks for updating the requirements, Sean.  Patch looks good to go.","accounts_in_message":[],"_revision_number":4},{"id":"836f8709ebb57a049631b5888df001ae7690b5f4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-04 14:45:30.000000000","message":"Patch Set 4: Code-Review+1\n\nActually, I should wait to see that the tests all pass.","accounts_in_message":[],"_revision_number":4},{"id":"389b7118b4323d489c45e44dd534fa1477d2c373","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-06-04 14:47:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/05/733105/4/check/fujitsu-eternus-dx-iscsi/791b7b5 : FAILURE in 35m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/05/733105/4/check/fujitsu-eternus-dx-fc/27fcd4e : FAILURE in 14m 08s","accounts_in_message":[],"_revision_number":4},{"id":"814c4b695f1d9478c92420863872d362c08b2263","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-04 15:03:13.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/733105/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"f821584d76d0c23ccd5f212bf9b73d5ecd78c4b2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-04 15:09:50.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/733105/4/iscsi/ : SUCCESS in 57m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/733105/4/smb/ : FAILURE in 56m 56s","accounts_in_message":[],"_revision_number":4},{"id":"8417e1375928e3c368864b0a98668eb945266abd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-04 15:21:50.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/733105/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/94b0013/94b0013.html : FAILURE in 1h 08m 42s","accounts_in_message":[],"_revision_number":4},{"id":"4bb72521c776b5534b14ab70a40584492e683b97","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 15:48:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"217e4e8756a90bf98741a86ae4b8632c8dcabed3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-04 16:30:55.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-733105-4/ns5_nfs : SUCCESS in 2h 19m 11s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-733105-4/ns5_iscsi : SUCCESS in 2h 13m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8837f12c2656f1756bbdce75d85974aac09a31c3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-04 16:31:49.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/733105/4/check/tempest-dsvm-full-bionic/d6b1666/ : SUCCESS in 1h 31m 29s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/733105/4/check/devstack-plugin-ceph-dsvm-bionic/fce4dc0/ : SUCCESS in 1h 49m 02s","accounts_in_message":[],"_revision_number":4},{"id":"030c7b6dd7bbd9fd0a684ef95b470a91507c34bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 18:28:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cee29f13d4284c47b3b1b27d483d95af : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7963bc525cb4e88ae25eff3819b6a11 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75ab99b732c64efbb6cee35fd282b246 : SUCCESS in 11m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bb6109137434c7db0f44533592d3013 : SUCCESS in 9m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11894463ab054c649dbdab54ef00f05f : SUCCESS in 15m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/12dd70b1ff0445c9934410aa33afbb92 : SUCCESS in 4m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f265013774274855b2c223ccc516dc80 : SUCCESS in 42m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/407efa50fda443ed840d794fcf4d52e1 : SUCCESS in 1h 07m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a906b9f1d934decb3ce2bc3fd3ce8ea : SUCCESS in 11m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dc56fe61a5245cba112c9060543a65f : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4a0327ce78240f4b3f9665cfff88833 : SUCCESS in 7m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec3320eca9034bb398533aa09ccde05f : SUCCESS in 7m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2dceb7a518bd457d852f0e7195216cf5 : SUCCESS in 46m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3691bdbf8c364cc6a35d4bfdc681cb91 : SUCCESS in 7m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/08ba3b7eec1349689ea2b286a6e4966f : SUCCESS in 1h 16m 45s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40844458a185493ba71f0fa3bff668a2 : SUCCESS in 1h 26m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/22fea9e165014b2f953c68eb80550690 : SUCCESS in 1h 08m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c106b046f45c4135b4f5b6e5279621e6 : SUCCESS in 1h 17m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/497a92eebfd447c5a96a91a95c5ac197 : SUCCESS in 1h 31m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc7e98af26df4f7d9b7887dbd2f22257 : SUCCESS in 2h 18m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6ec8000b0a9473096d716ac49ba44b0 : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":4},{"id":"0dde26b517590ca2ac070ae4e8a5d290671ace7b","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-06-04 18:47:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/05/733105/4/check/nimble-iscsi-driver-dsvm-volume/856a7fc : SUCCESS in 1h 42m 13s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/05/733105/4/check/nimble-fc-driver-dsvm-volume/645d5b2 : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":4},{"id":"23dbd02a2e82c200515f1952c387a10e95ad75e6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-04 18:58:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLooks good!","accounts_in_message":[],"_revision_number":4},{"id":"c88d2c7c96926d208a0905bd58657e0a287008bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 18:58:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8cc891f42588fd7270dc9f596c738c407202c71a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-04 19:12:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/05/733105/4/check/cinder-storpool-tempest/3223983/ : RETRY_LIMIT in 11m 27s","accounts_in_message":[],"_revision_number":4},{"id":"4a2ad13a89ecba461bb38d9122077006a76ee8dd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-04 19:49:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/733105/4/check/dsvm-tempest-cisco-zonemanager-job/f4c007b/ : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":4},{"id":"69479efa48fe8714d9d197e8d2d07a45c92033dd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-04 19:58:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/05/733105/4/check/DellEMC_Unity_FC/6f3bbca/DellEMC_Unity_FC/120 : SUCCESS in 1h 20m 04s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/05/733105/4/check/DellEMC_Unity_iSCSI/b8ca5bf/DellEMC_Unity_iSCSI/114 : FAILURE in 1h 43m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"67cd0763b361906b72b016ca2ca9e829f30dd595","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-04 20:00:37.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-733105-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"bc737a664162f082ccdf9bc07decce8041eaa544","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 20:24:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"7be74cb9cedaa0c0dd379b1412dbd46d8913c0a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 20:24:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adff10d46f2d4820bf19dd347cebc2cb : SUCCESS in 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa188da8587437386c0418c6b91576a : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8b6e55d7834c9a8ae70b73fa7537a8 : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a1f4db695343a0ab77caf551779396 : SUCCESS in 14m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a7013e765684136a2765d3b132b80c1 : SUCCESS in 12m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6e119a2f44c4b2bbc4141b12e519ed7 : SUCCESS in 4m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/b9db25cbe06c471086316a21b873c58c : SUCCESS in 43m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32c55f7e4b534cf78fb3a96c6d75e26a : SUCCESS in 59m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5b84a72c299462ea55bb79825b6c09f : SUCCESS in 9m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/45dbf70dd96a4ee88dfc246947edee28 : SUCCESS in 1h 18m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/179c7d9748e249ef921a93e9991d35ef : SUCCESS in 1h 15m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b992f47354844745ac7740f2f23ede01 : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":4},{"id":"826a6123ab877fb7dbfedb0afc91ecc3023e90c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 20:26:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1a80ef54614bea9413beb0de4b67cd : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/577a583baef84e719c13c8bd49bbb536 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":4},{"id":"7a1969776ce6f278353c5410f6d265b867279d5e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-05 01:08:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/05/733105/4/check/DellEMC_XtremIO_FC/72622ff/DellEMC_XtremIO_FC/129 : SUCCESS in 1h 31m 01s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/05/733105/4/check/DellEMC_XtremIO_iSCSI/ea63ca3/DellEMC_XtremIO_iSCSI/128 : FAILURE in 1h 58m 30s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"97db82a9c2d0921bfeaa41398193f5048321e9db","author":{"_account_id":31254,"name":"Rishabh Sikka","email":"rishabh.sikka@hpe.com"},"date":"2020-06-05 04:25:54.000000000","message":"Patch Set 4:\n\nrun-HPE Storage CI","accounts_in_message":[],"_revision_number":4},{"id":"bd0f120c06d43b58b6558ee15c67df6babfd118a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-05 09:07:34.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/05/733105/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/0b78018 : SUCCESS in 57m 59s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/05/733105/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/489d88d : FAILURE in 1h 53m 37s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"4d30e83003f5f4c9b00111096f380bd370705ba7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-05 09:10:36.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/733105/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"db717d4678c90ffaa7ff95d480db29b7916e6ce2","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-05 10:18:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1143/index.html : SUCCESS in 1h 16m 02s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1294/index.html : SUCCESS in 1h 47m 04s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1321/index.html : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":4},{"id":"5d1c97aeea98db991234e15e28261f0b944a1c76","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-05 13:08:12.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/05/733105/4/check/DellEMC_VxFlexOS/3b69a8c/DellEMC_VxFlexOS/94 : ABORTED in 4h 00m 31s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"d12be1190f1de1f2993abb819b6cdc5bfeed8806","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-06 02:37:34.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/733105/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3b85fbc1568e82609ede2b138ead5e89793ce61c","author":{"_account_id":31254,"name":"Rishabh Sikka","email":"rishabh.sikka@hpe.com"},"date":"2020-06-08 06:19:56.000000000","message":"Patch Set 4:\n\nrun-HPE Storage CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b2c9592281b24824c2158f91a9295288f41dd81a","revisions":{"5050902d482b8404e6c3375bd51b2d3364db5d3c":{"kind":"REWORK","_number":1,"created":"2020-06-03 11:14:03.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/05/733105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/733105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/733105/1"}}},"commit":{"parents":[{"commit":"307fccea4cc9c6496334b0fe137206cb48499bd5","subject":"Merge \"Add log if resource_filters json does not exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/307fccea4cc9c6496334b0fe137206cb48499bd5"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-01 08:32:38.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-03 07:28:03.000000000","tz":180},"subject":"Remove VxFlex OS credentials from connection_properties","message":"Remove VxFlex OS credentials from connection_properties\n\nVxFlex OS password is not stored in block_device_mapping table. Instead of this\npasswords are stored in separate file and are retrieved during each attach/detach\noperation.\n\nCloses-Bug: #1823200\nChange-Id: Ia7cc17472677a693c6162f0b8b0529df62eed7cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5050902d482b8404e6c3375bd51b2d3364db5d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5050902d482b8404e6c3375bd51b2d3364db5d3c"}]},"branch":"refs/heads/master"},"52be8aff00a3debe90811d6e9cad7032f075b026":{"kind":"REWORK","_number":2,"created":"2020-06-03 12:25:20.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/05/733105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/733105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/733105/2"}}},"commit":{"parents":[{"commit":"307fccea4cc9c6496334b0fe137206cb48499bd5","subject":"Merge \"Add log if resource_filters json does not exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/307fccea4cc9c6496334b0fe137206cb48499bd5"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-01 08:32:38.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-03 12:25:10.000000000","tz":0},"subject":"Remove VxFlex OS credentials from connection_properties","message":"Remove VxFlex OS credentials from connection_properties\n\nVxFlex OS password is not stored in block_device_mapping table. Instead of this\npasswords are stored in separate file and are retrieved during each attach/detach\noperation.\n\nCloses-Bug: #1823200\nChange-Id: Ia7cc17472677a693c6162f0b8b0529df62eed7cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52be8aff00a3debe90811d6e9cad7032f075b026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52be8aff00a3debe90811d6e9cad7032f075b026"}]},"branch":"refs/heads/master"},"39538f2d1f48f867a795172a8e1de9673d6cb1d3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-04 05:09:48.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/05/733105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/733105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/733105/3"}}},"commit":{"parents":[{"commit":"ce5709d0e27413b56d8a645d9e6c9bfe46267035","subject":"Improve tenacity retry sleep mocking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce5709d0e27413b56d8a645d9e6c9bfe46267035"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-01 08:32:38.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-04 05:09:48.000000000","tz":0},"subject":"Remove VxFlex OS credentials from connection_properties","message":"Remove VxFlex OS credentials from connection_properties\n\nVxFlex OS password is not stored in block_device_mapping table. Instead of this\npasswords are stored in separate file and are retrieved during each attach/detach\noperation.\n\nCloses-Bug: #1823200\nChange-Id: Ia7cc17472677a693c6162f0b8b0529df62eed7cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39538f2d1f48f867a795172a8e1de9673d6cb1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39538f2d1f48f867a795172a8e1de9673d6cb1d3"}]},"branch":"refs/heads/master"},"b2c9592281b24824c2158f91a9295288f41dd81a":{"kind":"REWORK","_number":4,"created":"2020-06-04 14:11:23.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/05/733105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/733105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/733105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/733105/4"}}},"commit":{"parents":[{"commit":"ce5709d0e27413b56d8a645d9e6c9bfe46267035","subject":"Improve tenacity retry sleep mocking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce5709d0e27413b56d8a645d9e6c9bfe46267035"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-06-01 08:32:38.000000000","tz":180},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-04 14:11:18.000000000","tz":-300},"subject":"Remove VxFlex OS credentials from connection_properties","message":"Remove VxFlex OS credentials from connection_properties\n\nVxFlex OS password is not stored in block_device_mapping table. Instead of this\npasswords are stored in separate file and are retrieved during each attach/detach\noperation.\n\nCloses-Bug: #1823200\nChange-Id: Ia7cc17472677a693c6162f0b8b0529df62eed7cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2c9592281b24824c2158f91a9295288f41dd81a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2c9592281b24824c2158f91a9295288f41dd81a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
