)]}'
{"id":"openstack%2Fcinder~736805","triplet_id":"openstack%2Fcinder~stable%2Fstein~I018007bd865b69631c21af83efdf6aba981161f1","project":"openstack/cinder","branch":"stable/stein","topic":"ossn-0086-followup","hashtags":[],"change_id":"I018007bd865b69631c21af83efdf6aba981161f1","subject":"Update os-brick library to 2.8.6","status":"MERGED","created":"2020-06-18 19:11:16.000000000","updated":"2020-07-01 14:12:46.000000000","submitted":"2020-07-01 14:10:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736805-1593612644860-b4f81718","meta_rev_id":"c4487925e6b50f26dda6fe6c5ea641328adbed11","_number":736805,"virtual_id_number":736805,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-29 23:40:59.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-30 22:44:30.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2020-07-01 02:32:21.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-29 22:21:10.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-06-30 00:19:57.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-06-30 21:46:28.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-01 14:10:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-01 00:02:00.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-30 00:41:20.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-01 04:08:29.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-07-01 01:28:35.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"date":"2020-07-01 01:38:05.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-01 10:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-01 10:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-01 10:46:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-01 10:52:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":2,"date":"2020-07-01 00:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-20 00:02:31.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-06-29 22:21:10.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-06-29 23:40:59.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 00:19:57.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 00:41:20.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 21:46:28.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-06-30 22:44:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-07-01 00:02:00.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 01:28:35.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 01:38:05.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-07-01 02:32:21.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 04:08:29.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-07-01 10:46:53.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-07-01 10:52:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-01 10:52:59.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-07-01 14:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94bb07b6b4bcdce99723650b629c51b7f966df5d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-18 19:11:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7e0e02532e89330f969a425cfc5e3747dde5c48","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-18 19:12:18.000000000","message":"Patch Set 1:\n\nThe upper-constraints change that will allow this hasn\u0027t merged yet, but I could use some feedback on the release note.","accounts_in_message":[],"_revision_number":1},{"id":"0a94b9fdef0074aad66ea4ad333cda594a42242a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-18 19:19:17.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/736805/1/iscsi/ : FAILURE in 7m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/736805/1/smb/ : FAILURE in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"439ddaf3de0d34df41b0467537cdd0c103c0609f","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-18 19:23:08.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/00b1818979aba8e9abd5bef509a74fb27b5602af/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0cdaa170bbe901365627cbcea40db4b54b38c59e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-18 19:39:56.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/05/736805/1/check/dsvm-tempest-infinibox-fc/ef34c20 : FAILURE in 28m 10s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/05/736805/1/check/dsvm-tempest-infinibox-iscsi/b204b71 : FAILURE in 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2556133bf5745320de2de835a6566569237f7595","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-18 19:44:22.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"458df52000f1556120771091e971de7773e86cd4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-18 19:51:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/736805/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f730703/f730703.html : FAILURE in 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b464286739f02fe789bcbc01c14032f5f96f53a9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-18 19:55:33.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b1ee7d0f7662c1a65989b61725c804527053b424","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 20:07:36.000000000","message":"Patch Set 1: Code-Review+2\n\nWe could mention in the release note that it only impacts py27 deployments, but I think the link to the os-brick release note is good enough.","accounts_in_message":[],"_revision_number":1},{"id":"f49ba624d3555a156a9a6d2f764e7dccdf8a17f7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-18 20:19:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/736805/1/check/dsvm-tempest-cisco-zonemanager-job/4ff6f89/ : FAILURE in 1h 02m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8b82252ee2dc390c77b06de8e6dc7355430327c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 21:21:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1fd2cf43004422c9fa79739c5d599f1 : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21bdc4fc9f0448588e38554a9f84687b : SUCCESS in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d93376c9319a42b2b642407f008d7429 : FAILURE in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59c870e1d3844992a972316edd07a0fc : SUCCESS in 13m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e52b87818e8450182ecd68645356b33 : SUCCESS in 10m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8fc9b8a6bda4788aa6a88798944aa1b : SUCCESS in 3m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/16e022a732444abf87582d5d09d102f3 : SUCCESS in 1h 26m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e72c02af12b146008c153b5ac6d19991 : FAILURE in 20m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbe3a09dd9d14907abda69a18913372a : SUCCESS in 1h 30m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6bc9939c898458fa51adceb9569a7a1 : SUCCESS in 12m 51s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0592331fb7e942d693544aed4b8e5035 : SUCCESS in 1h 19m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c01bdb99a994040821a63a765371565 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d391c5808744d389f240ec7641f71b1 : SUCCESS in 7m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/88e88ec65c5343c48680ed1bb06deeca : FAILURE in 16m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4935520e003a41bba6ff8263c11c710a : SUCCESS in 7m 58s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ceee3f50a0d4fdd82cebc62c3543279 : SUCCESS in 1h 15m 35s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8fcd61c5e10e4becbe21fd480870c70e : SUCCESS in 6m 32s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9a0af07c49244fc0bb3425d55a1a94b9 : FAILURE in 19m 12s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9aae6f5dfbe843b2b594ac1ce18e61e1 : SUCCESS in 1h 19m 21s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/6d7eab43a1a848a485b569a456fcba98 : SUCCESS in 1h 35m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a105e94dd845446291902ebf617039be : SUCCESS in 2h 04m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c313b33af0cf4209adbe07f222479f29 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"19d87abc0c2817fb80b74766f54c7d427aa36ff1","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-18 21:51:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/05/736805/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/6ccb463 : FAILURE in 1h 03m 57s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/05/736805/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/26520be : SUCCESS in 1h 08m 35s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"c1db132da886d40dc3cd797400d75289a7c3665c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-18 23:07:47.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/1/check/tempest-dsvm-full-bionic/ebda9ce/ : TIMED_OUT in 2h 37m 51s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/1/check/devstack-plugin-ceph-dsvm-bionic/fb605e8/ : FAILURE in 1h 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cfc2e5c10ad391046157a028e4b760373bb5beec","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 00:56:18.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d01ebf6c6f435ccb783810f281eed9d3a065870b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-19 11:10:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8addc15e93fcdea5f9d3565db0b2e6fc1249858d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-19 11:18:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/736805/1/iscsi/ : FAILURE in 7m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/736805/1/smb/ : FAILURE in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9b3da2f84acd51b9b79755fec248489d79e784cc","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 11:44:39.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ffdcd4a6d4819f96ffcf13d5b3137bf199b099dd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-19 12:18:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/736805/1/check/dsvm-tempest-cisco-zonemanager-job/7dbdc2d/ : FAILURE in 1h 01m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e026f63d10b6b4564f800507d4d5658a812c35c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:38:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a217345ae0949dc970070a22b0fdf6d : SUCCESS in 21m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c2e3e974814af68e663234091f5bd6 : SUCCESS in 14m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f0ae9ad2b6446f7a058c3d97900e1ea : FAILURE in 12m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c919f040156a4793aa5d94978215c7ac : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df36500b1187454c943f865ac41d478f : SUCCESS in 15m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/43bd8389aeb545f8bd8a71a729d1c0d9 : FAILURE in 7m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e76bcbda72ad4ad0bb7a1456d82e270b : SUCCESS in 1h 00m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dd0f8fe048a641698bad88ed2076d295 : FAILURE in 23m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3a67910fed1450a85a7212fcaa51517 : SUCCESS in 53m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6616389094ad4486b1ffa327b4070c5c : SUCCESS in 7m 53s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/70d75a014a2c42deb29658069d5bbe15 : SUCCESS in 1h 11m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc690444c6b54d70a2525ba5f1549442 : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef7a496375784876a601ffff1bd32d33 : SUCCESS in 7m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b3adc6f1e80d41ca9da2a8b757762ef6 : FAILURE in 16m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e92cdd6f6bd4aa7aa415b0bcff45785 : SUCCESS in 7m 17s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/280952a5573045c183dfbeecb9c9c244 : SUCCESS in 56m 49s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4c1f9e72648743f5970353fe0f17a904 : SUCCESS in 7m 44s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f8ba1eeef5984da39ed3715682c82c13 : FAILURE in 12m 48s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/25e5b3613d824c0ba9ba26cceedd0cbd : SUCCESS in 1h 15m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/1417bfd6f5d74b4c862a9156ef469456 : SUCCESS in 1h 22m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fbccf464ccee4297b99c28dd49a81f9d : SUCCESS in 1h 26m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cb42ab0976844bf28982041756460eb2 : FAILURE in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ea24a5ee8809e6eafb41c76305edd516712988d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 13:29:04.000000000","message":"Removed Code-Review+2 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"2b0139fe32cb4daa61bbe1b9dbd000e7bbf65bb6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 13:29:39.000000000","message":"Patch Set 1:\n\nPy2 needs six:\n\nNote The ConfigParser module has been renamed to configparser in Python 3. The 2to3 tool will automatically adapt imports when converting your sources to Python 3.","accounts_in_message":[],"_revision_number":1},{"id":"8640647cf711b0e148af0476fbaa5e60761d06fd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-19 13:36:20.000000000","message":"Patch Set 1:\n\nos-brick 2.8.6 is using configparser from six:\nhttps://opendev.org/openstack/os-brick/src/tag/2.8.6/os_brick/privileged/scaleio.py#L20\n\nLooking at the tox log, it looks like os-brick 3.0.2 (which is py-3 only) is being used.","accounts_in_message":[],"_revision_number":1},{"id":"66a8e2538105739e4ab2445d6cd81040dd3b14da","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-19 13:42:09.000000000","message":"Patch Set 1:\n\nBTW, the upper-contraints change hasn\u0027t merged yet: https://review.opendev.org/#/c/736784/","accounts_in_message":[],"_revision_number":1},{"id":"f5eb5db8440be62593843b9b704135a4d7b0d3a8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 13:42:27.000000000","message":"Patch Set 1:\n\nSomehow this job is using master upper-constraints:\n\npy27 installdeps: -c/home/zuul/src/opendev.org/openstack/requirements/upper-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"0dff53eaeb44325494e5b2b0ff2d722482485818","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 13:43:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dc25d784eba471908c299f14f798ca8345903f3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 13:45:34.000000000","message":"Patch Set 1:\n\n2020-06-19 11:13:38.617659 | ubuntu-bionic | opendev.org/openstack/requirements checked out to:\n2020-06-19 11:13:38.617851 | ubuntu-bionic | d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c update constraint for oslo.log to new release 3.42.5\n\nThat commit is the correct HEAD of stable/stein requirements.","accounts_in_message":[],"_revision_number":1},{"id":"5b2c940d3ad00b7e4aa34d8a3ab1c21da9229081","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-19 13:50:49.000000000","message":"Patch Set 1:\n\nThe pip freeze from the py36 job looks suspicious, too:\nhttps://95ec6d5180d45c332d30-3a495235250eccc64faa024f22d41fdc.ssl.cf1.rackcdn.com/736805/1/check/openstack-tox-py36/c919f04/tox/py36-3.log\n\nit\u0027s using os-brick 3.1.0","accounts_in_message":[],"_revision_number":1},{"id":"d685dc74c55d705028fbbc65f97d093ae2fff1f7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-19 13:53:37.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/1/check/tempest-dsvm-full-bionic/4c780bb/ : TIMED_OUT in 2h 36m 22s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/1/check/devstack-plugin-ceph-dsvm-bionic/115ae40/ : FAILURE in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8830fd307074c26f259e358282a2a78bff91429d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-19 14:29:16.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a30b48714fedf0f1a12949c5f07c0a689e9dc1a2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-19 14:40:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/05/736805/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/07ea72b : FAILURE in 1h 33m 00s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/05/736805/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/7b8023c : SUCCESS in 1h 19m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"010ded9e48f1afa71f5b62e20a070a68990129ab","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-06-20 00:02:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1231/index.html : SUCCESS in 1h 15m 34s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1378/index.html : SUCCESS in 1h 37m 13s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1403/index.html : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2681885924b199aa57e283447c80d8496fe8155a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-23 21:54:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2d1c07b38c54126293c068e2c0024de8a10ab1fb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-23 22:03:05.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/736805/1/iscsi/ : FAILURE in 7m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/736805/1/smb/ : FAILURE in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5fd7c2daf52069ef60d774eaa9f748d35e62d002","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-23 22:37:27.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"912e2558f619776eb9de8ae59a55b42821965b3c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-24 01:55:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/736805/1/check/dsvm-tempest-cisco-zonemanager-job/c240690/ : FAILURE in 1h 01m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b4b7418b4738c4d785828ab199ed3cbcf2086fc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 03:26:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f384adfb69e640f58076d804b94b0574 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe6980f8c384133b69387cbcce2508b : SUCCESS in 9m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6184853811643828d98a109cf9f8f10 : FAILURE in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c66f8f7d59344a71ac43f21fba6cd68e : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec73d65e9394f9daf2c123e1b13752a : SUCCESS in 9m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9586124e8f404234abc7324147167387 : FAILURE in 4m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3ee83a777424441fb462259a46382ee2 : SUCCESS in 1h 32m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/493102fd22ce4d798f601fd5de21b9f4 : FAILURE in 18m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc97c76e486149ccb7b24d2e211e15e5 : SUCCESS in 1h 25m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7925ab9100eb474ebaf4857cb2115a4e : SUCCESS in 7m 35s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5bafb0f3ea6e4116854b86dd8a11beaf : SUCCESS in 1h 22m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95679df46a99414fbf48c567f0bd2f29 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71987f2c96f64e15a51097aa74e0c4f5 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3744d8537aad4b018f9351e3c7f540cf : FAILURE in 20m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f70d8b0b80a84f47b72071d3b8f32002 : SUCCESS in 5m 15s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/99f99b77898f45239143d59440c76e43 : SUCCESS in 1h 04m 38s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e023138129274423a12c4064a09e6508 : SUCCESS in 5m 37s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ec9e3c93f6341d395d8be2422f28086 : FAILURE in 20m 40s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dc05f214af5e4907af7d019d3ab80cc9 : SUCCESS in 57m 11s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/9cf2b18b113f4178b53084288388aed5 : SUCCESS in 1h 02m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e5600e824124544895d400723f9ec30 : SUCCESS in 2h 02m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ed298fc4a76c4467898f72ce5048538f : FAILURE in 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"cf49a7b95ab0f96171228a613bc947105a074bc9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-24 04:24:20.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/1/check/tempest-dsvm-full-bionic/e79a04e/ : TIMED_OUT in 2h 36m 56s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/1/check/devstack-plugin-ceph-dsvm-bionic/b77c4b1/ : FAILURE in 1h 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"23cda3c4b9cedd730b4f1e8191072edd5a8dd827","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 05:29:23.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d18d26f620ceb23dd182736e1f4351ae5c9bbf1f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-24 10:39:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/05/736805/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/80f1ca6 : SUCCESS in 1h 00m 43s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/05/736805/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/71f1e78 : SUCCESS in 1h 08m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"b39aafd9d4a4398a3df529c0049fe2c630c057ab","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-24 12:25:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bfe75ff75bc431992e12b25ef9f669a2d79a79af","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-24 12:34:32.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/736805/2/iscsi/ : FAILURE in 8m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/736805/2/smb/ : FAILURE in 8m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2f3051657da46d39a712fb606ba355d477f3775f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-24 12:55:06.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/05/736805/2/check/dsvm-tempest-infinibox-fc/c0da73b : FAILURE in 28m 36s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/05/736805/2/check/dsvm-tempest-infinibox-iscsi/2168dbe : FAILURE in 21m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3b6056e1b333d3bfc179a95961298e157bcf0c30","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-24 13:02:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/736805/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/582ad65/582ad65.html : FAILURE in 35m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f2ae68050aa23bf48881007f23dfaa03084cde88","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 13:22:59.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"fc2eaba68a4a18ebe2a795e3ccd42df707fe3720","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 13:25:33.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"73dd7232ff813b4e2698b2ed5a93f728957eeb4d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-24 14:29:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/736805/2/check/dsvm-tempest-cisco-zonemanager-job/95a9406/ : FAILURE in 1h 00m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b663a3a02ac05fcccb2e2871528f3b4176d35c9e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-24 15:36:13.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/2/check/tempest-dsvm-full-bionic/b868734/ : TIMED_OUT in 2h 39m 15s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/2/check/devstack-plugin-ceph-dsvm-bionic/e3fd0f0/ : FAILURE in 1h 09m 03s","accounts_in_message":[],"_revision_number":2},{"id":"32e448369d165ba8e28a30652b8e62cb544c7bd4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-24 17:03:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/05/736805/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f192ca6 : SUCCESS in 1h 06m 24s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/05/736805/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/8b2dcf0 : SUCCESS in 1h 17m 25s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"d36a55ad0fc68a0542dab21c2357a7cbfd887442","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 17:44:25.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d00090c3843c5ea1096769f29678c1f5d86d2266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:03:08.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"ee9d8665346e8d47bd30fdc2b14a8193f861ce53","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-25 20:50:03.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2e043fad4b4ec511c062a43710220f1c246429f8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-25 20:58:43.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/736805/3/iscsi/ : FAILURE in 8m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/736805/3/smb/ : FAILURE in 7m 54s","accounts_in_message":[],"_revision_number":3},{"id":"182ce9a55b30df9af5d2e29c4f2741d1f3c6523c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-25 21:18:34.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/05/736805/3/check/dsvm-tempest-infinibox-fc/6c0bfce : FAILURE in 27m 57s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/05/736805/3/check/dsvm-tempest-infinibox-iscsi/52ea1da : FAILURE in 21m 23s","accounts_in_message":[],"_revision_number":3},{"id":"49d71425544e7a5ad11c97b6609b366ec6d9b9c6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-25 21:25:26.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"f48203c6409861d2d2d17eb3eba2a1a4444afab4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-25 21:32:07.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"556d5d8bf53752007cf822702444b6e6ecf59b2e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-25 21:38:20.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/05/736805/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/66dffed/66dffed.html : FAILURE in 46m 58s","accounts_in_message":[],"_revision_number":3},{"id":"7985a4722f33f35c8700284ea4f737464d76fa1a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-25 23:37:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/736805/3/check/dsvm-tempest-cisco-zonemanager-job/d5b2455/ : FAILURE in 1h 01m 51s","accounts_in_message":[],"_revision_number":3},{"id":"e7311c7501e92e9076961925fa1a08fbdf71c704","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-26 02:18:04.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/94166062803247a1af486d0697483f8d : RETRY_LIMIT in 21s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/3/check/devstack-plugin-ceph-dsvm-bionic/560f3d3/ : FAILURE in 1h 09m 59s","accounts_in_message":[],"_revision_number":3},{"id":"b2cdeee517d4527b3012049740a5a911736fad66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 02:35:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1356dd222f864955a50e7153e9da0905 : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e24ec6cd0044eccb0961bd2171a4a22 : SUCCESS in 7m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a16e10939e34c98bbabcea72d77a231 : SUCCESS in 11m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f224fee6890421daccfaaaee3f0e4ca : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/500090527da94cefa3233fc722dfb9b4 : SUCCESS in 11m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/66061fde58ef40c7b809ae25241a4405 : FAILURE in 3m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9176369ca8994ca69c78444d9de3239c : SUCCESS in 1h 33m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6e154aedc02c4c0dad0aa40748f8acea : FAILURE in 20m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a3dddb42cb84aa9ab4249b84b011cfa : SUCCESS in 1h 27m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b436cf2b73364d9fb66198b526fca516 : SUCCESS in 7m 49s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6d8ec2a8fab43d29c3c7cc74950d41d : SUCCESS in 1h 34m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd5b7bcda6334886b1d3d1306dd04fcb : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8518135268a49d1bb88be8174d6034b : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/693b5ee45dc74dc3b01f279c9ac913e2 : FAILURE in 20m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7769ef9cfeef46c1874f028d5ebf17b1 : SUCCESS in 5m 05s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c1d1ea7c574b4c919436b54ee24c8762 : SUCCESS in 1h 03m 11s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1fd059ce58794253b8e120fe8974bb68 : SUCCESS in 5m 50s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eb5a26f6a7024cc9b74cb51210f7dd29 : FAILURE in 26m 07s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a732913c6b9d45c3853ca7340d0465a3 : SUCCESS in 1h 19m 01s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/b3815b8b05eb499180a44ef891d5a43e : SUCCESS in 1h 27m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb6fc0c3fa1d4fe79fcee3999b8060d0 : SUCCESS in 2h 27m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5ca163676a8342e0b21253a28d0b082c : FAILURE in 21m 34s","accounts_in_message":[],"_revision_number":3},{"id":"3fe86ed13b8c7a967f9adff57e6fd09ec10f1e73","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-26 03:20:05.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"97c9a7ec830c3b7c7b3ba57e545d477f813c8d6a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-26 05:25:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/05/736805/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/79a013d : SUCCESS in 1h 02m 42s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/05/736805/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/453c3dc : SUCCESS in 1h 16m 40s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"c03516cd7addf71dc10ed5e44e03e6f02e9fb95c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-29 13:25:39.000000000","message":"Patch Set 3:\n\nNeed https://review.opendev.org/#/c/737408/ to merge to cinder stable/stein first.\n\nrequirements-check job is failing because it can\u0027t find virtualenv","accounts_in_message":[],"_revision_number":3},{"id":"5765b812511c98bb1441b2f498b3957e72a046b4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-29 21:22:26.000000000","message":"Patch Set 3:\n\nhttps://review.opendev.org/#/c/737408/ has merged.","accounts_in_message":[],"_revision_number":3},{"id":"9230cb2085eb35e0b05d80c55ba92aab25c0a5ce","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-29 21:22:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"52b092c9c7e4c9f7bfa887039d50544f52338686","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-29 21:28:40.000000000","message":"Patch Set 3:\n\n\u003e https://review.opendev.org/#/c/737408/ has merged.\n\nBut the requirements-check is going to fail again.","accounts_in_message":[],"_revision_number":3},{"id":"e9421eaef84c6c4970033117f90b7806fd2c105e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-29 21:31:04.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/736805/3/iscsi/ : FAILURE in 7m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/736805/3/smb/ : FAILURE in 8m 05s","accounts_in_message":[],"_revision_number":3},{"id":"54b3cd17cace3e1d3fc7a6439325e647d92bde47","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-29 21:35:53.000000000","message":"Patch Set 3:\n\nI guess we need the ensure-virtualenv role added to the requirements job in stable/train and stable/stein?","accounts_in_message":[],"_revision_number":3},{"id":"2a7a60ea6ae0658f4a52fe3bbbfe4dda6afb2c97","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-29 21:39:28.000000000","message":"Patch Set 3:\n\n\u003e I guess we need the ensure-virtualenv role added to the\n \u003e requirements job in stable/train and stable/stein?\n\nMaybe, but I don\u0027t really know how that job works.","accounts_in_message":[],"_revision_number":3},{"id":"10ef1c7dfc50b181dcd18ff6a95a63f8ba7ddd19","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-29 21:45:22.000000000","message":"Patch Set 3:\n\nMaybe https://review.opendev.org/738518 will do it.","accounts_in_message":[],"_revision_number":3},{"id":"b47871535297303544064a7f7f2c86196deb9c19","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-29 21:45:54.000000000","message":"Patch Set 3:\n\n(Well, for Train, then it will have to be backported to stein before it does us any good here.)","accounts_in_message":[],"_revision_number":3},{"id":"bd2388e7e5bb413e9c8cd66c9a17fbf4f00864fd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-29 21:51:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"17fa61e5fec6d092146344e99efd5b6e239e8e6d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-29 21:59:37.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/736805/4/iscsi/ : FAILURE in 7m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/736805/4/smb/ : FAILURE in 7m 29s","accounts_in_message":[],"_revision_number":4},{"id":"e3bf4db15f96309965fa7b18062b4043b677f1c6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-29 22:00:21.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"8e18f0a09964cd877b73b147a7c8f1d121894e73","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-29 22:21:10.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/05/736805/4/check/dsvm-tempest-infinibox-fc/c894454 : FAILURE in 28m 58s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/05/736805/4/check/dsvm-tempest-infinibox-iscsi/5d5708a : FAILURE in 21m 10s","accounts_in_message":[],"_revision_number":4},{"id":"7ba5578c3d2f63513de0c83c0cd7c2721bfc5a7e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-29 22:54:19.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"27c585d51d3122e9a70007806236772821c1438a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-29 22:58:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/736805/4/check/dsvm-tempest-cisco-zonemanager-job/7a8d4bc/ : FAILURE in 1h 02m 07s","accounts_in_message":[],"_revision_number":4},{"id":"97338fa01fb571c85ea5c5d7ea0fc2853ca9d127","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-29 23:33:04.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"5bdfd6736c35137d6570789ff1a668a946dc22be","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-29 23:40:59.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/a14fd58f5dc1e4027e4148dfb1c40253ebfda8a3/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"0e0105e0689579b29f477db8955c4b120bdb6a1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 23:55:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02f20a05a8ad4e4cbc2a27b1fcf5a19d : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f06916d715d49ed87d057062981253d : SUCCESS in 10m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fcd5798dc3647a9b4e9e9e17fba663a : SUCCESS in 11m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65d2e559150a451a930f891c3ecd0857 : SUCCESS in 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe683f482764a24b9bc6eb0e02744c3 : SUCCESS in 7m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b9acab5e18248c384c9d97008d4b54a : FAILURE in 5m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4ff44618b59e4af9a56430f5aa53b34a : SUCCESS in 1h 35m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5482916993f1428bb00267a0fca49c3f : FAILURE in 17m 15s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec934b7fc6ab4e43a143b7a489c04c7f : SUCCESS in 1h 25m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adc5319ae7514d7c9318b69905ca5fdd : SUCCESS in 9m 35s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b2ec7d618884a369729b922f8869e36 : SUCCESS in 1h 32m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ae0413cd6a64aae8377bd14135d233b : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4781b3a0157146108136a857ac999bb9 : SUCCESS in 10m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca9f4ed11bf84c28b73db55e12d90a2a : FAILURE in 19m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7fb692d78064ea2b6c297ca9d2b40dc : SUCCESS in 10m 47s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aee6b5b30cb1475aa312bade05d1dfce : SUCCESS in 55m 16s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cdd4d7322a164168b02d81c2f31b4a83 : SUCCESS in 9m 26s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf5c560e2ce04480a10471466a0f67db : FAILURE in 21m 24s (non-voting)\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a99c29a2f8dd497f9fd179e2b32582f8 : FAILURE in 1h 24m 44s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/934d4378a3234de0bd1f9197aec176d5 : SUCCESS in 58m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1dfdc209a42a40b3a1ce69a808fd3b2b : SUCCESS in 2h 01m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/03f4d6319788498d983c900436f0156e : FAILURE in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cbd79df8cee9a8b2807b3c7d7449cf4766987e25","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-30 00:19:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 2h 27m 41s","accounts_in_message":[],"_revision_number":4},{"id":"e7c8afc6976b15739e80b9e83f8b5dbdba4fb659","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-30 00:33:08.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/4/check/tempest-dsvm-full-bionic/292aeeb/ : TIMED_OUT in 2h 37m 34s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/4/check/devstack-plugin-ceph-dsvm-bionic/a5419f7/ : FAILURE in 1h 06m 12s","accounts_in_message":[],"_revision_number":4},{"id":"97ffe478214f09a6c597ad2cd9c6fb2ed88ed796","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-30 00:41:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/05/736805/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/491b054 : SUCCESS in 1h 02m 52s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/05/736805/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/115af3a : SUCCESS in 1h 09m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"d7d55a8c9585b5943b88371eff5ecab58fcafdab","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 01:25:59.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"86f6ba6382dc47db64873c175909caedfe92996f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-30 21:36:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"03082e0f42620fa09390f7c78fe051884ea0bff1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-30 21:46:28.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/736805/4/iscsi/ : FAILURE in 9m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/736805/4/smb/ : FAILURE in 9m 11s","accounts_in_message":[],"_revision_number":4},{"id":"dee7e3d56cdf59e54b61a1d94b3d33c8c0e43d63","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 22:41:51.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"4f5737417755dae208962e2eccab77a5a7323d18","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-30 22:44:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/736805/4/check/dsvm-tempest-cisco-zonemanager-job/8239082/ : FAILURE in 1h 02m 53s","accounts_in_message":[],"_revision_number":4},{"id":"263ebb3160d8c6a9db55db7fe591db74e750ad76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 23:58:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca9e11b3d36f4e2c8ad09a3187599582 : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d6e8694abc546beaca3b52f1d17ef7e : SUCCESS in 8m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f20cf762b154cd3b4308e2b62dab37b : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ea2effc7e0e49a5b697a24a964f0155 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96711df2a9b435c95a795828571b504 : SUCCESS in 12m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/024e502c09aa428e9e412617798c8760 : SUCCESS in 5m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed6601be24364d0f8583064936c4820e : SUCCESS in 1h 30m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/38234931a0aa41dab69819687de48809 : FAILURE in 20m 43s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a82d944a0cc944aa956556f0be4340a8 : SUCCESS in 1h 19m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fed2930c9a748d9b89d2e46c46d96a1 : SUCCESS in 9m 13s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ed9114454e11433ab354833bafe882e0 : SUCCESS in 1h 55m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e362653ea504b9694f784e0dd77434f : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5cc3dcdf5a154984b6d20df7a85f3cd7 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/99506e5fd45e45cd87780e038fc94930 : FAILURE in 18m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9eeb0f63e1b64e179dc60c9d0cae4645 : SUCCESS in 6m 00s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/592d611cc53248cf8e53df5d3d28520f : SUCCESS in 1h 03m 14s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ce257ef9849b444eb187295817933e23 : SUCCESS in 6m 05s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/83fa5d539c0c41a4a05fba5f89e83303 : FAILURE in 22m 05s (non-voting)\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/088f2c5949204fe38d7b137dd66f71b6 : FAILURE in 1h 50m 20s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/3fff22b071e84c559b78e270afc665f9 : SUCCESS in 1h 19m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/719d6b198c814438aa00951bfce6d0e4 : SUCCESS in 2h 19m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d7dde39e723740f4be8af6eed6a8c70d : FAILURE in 20m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"df5113f32ac66ac1844bc0d3d187dabc64277544","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-01 00:02:00.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/05/736805/4/check/synology-dsm-cinder-iscsi/9accd3f : FAILURE in 3m 01s","accounts_in_message":[],"_revision_number":4},{"id":"11ade6d9c5b35b68dd3948b7a3df9172831f8c70","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-01 00:34:52.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"9537126ddbc3c7eb86c69828cc0eb8b5587bf2c4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-01 01:28:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/736805/4/upstream-check/cinder-cDOT-iSCSI/7faa9a3/ : SUCCESS in 44m 18s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/05/736805/4/upstream-check/cinder-cDOT-iSCSI-IPv6/ec8d959/ : SUCCESS in 1h 25m 22s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/736805/4/upstream-check/cinder-cDOT-NFS/8fa7746/ : SUCCESS in 42m 09s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/05/736805/4/upstream-check/cinder-cDOT-NFS-IPv6/3f5769a/ : SUCCESS in 1h 29m 13s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/736805/4/upstream-check/cinder-cDOT-FCP/fa07bb4/ : SUCCESS in 57m 06s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"4721984ca0a4884f52b166e1ad8681e0a0e43a78","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-01 01:38:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/05/736805/4/upstream-solidfire-check/cinder-solidfire-iSCSI/67d8b34/ : FAILURE in 1h 38m 26s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"382ed486ed295650d4dc5c845ef5acde8a805283","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 02:10:31.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"7e91210b593689c785c09dce60cac9039d39b300","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-01 02:32:21.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/4/check/tempest-dsvm-full-bionic/5acb487/ : TIMED_OUT in 2h 38m 54s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/736805/4/check/devstack-plugin-ceph-dsvm-bionic/a819527/ : FAILURE in 1h 06m 12s","accounts_in_message":[],"_revision_number":4},{"id":"932a36f189062219337e4541d5500bd3ea756ba5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 04:08:29.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/05/736805/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"2f1fef83693a720f2212c055b9aececc7607bd99","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-01 10:46:53.000000000","message":"Patch Set 4: Code-Review+1\n\nyay","accounts_in_message":[],"_revision_number":4},{"id":"f5b957d9a3dd9bc6f924e9c2fa6ec3f52fcb3aee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-01 10:52:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"df54ce54191ed737c8656484d853ca523d572fb4","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-01 10:52:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"930196b1f5ceee6f9f97eb29795a4e0b2642ea73","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-01 10:52:59.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8142eb73f93bf5f54bcc725372456e7425b287fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 10:53:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c9504989947faf295fce0d6dc433fc0af42f118d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 14:10:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e04cd1680544065b7ba616e80c1af96 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27713cb1462a48ab8545d6516a1cc47d : SUCCESS in 9m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6728e9d0cc9489bb1176abb9a0a308b : SUCCESS in 10m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4e5c2fbc437408f88b27fb4a3366ec9 : SUCCESS in 18m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d3e7bf67e5a4bc688d4d207f4d8752b : SUCCESS in 12m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eef187fa44e54d29844e89469d46ff30 : SUCCESS in 6m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/442a01284db44679895e6d56c8e013d1 : SUCCESS in 1h 24m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efeac9d0137d400caafa015701fcdfea : SUCCESS in 1h 28m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b57a3a201384625b852018fe0439626 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b3e0785f3192e7c16f043bbd3ca45c5d96c02b2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 14:10:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1b6c9646677eed9e994a58d69d42eabdddd4d713","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 14:12:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c2e0e631cc42ac89c42e5865e3943f : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a74285030a5e435b88b20f57e89811c2 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a14fd58f5dc1e4027e4148dfb1c40253ebfda8a3","revisions":{"00b1818979aba8e9abd5bef509a74fb27b5602af":{"kind":"REWORK","_number":1,"created":"2020-06-18 19:11:16.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/736805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/736805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/736805/1"}}},"commit":{"parents":[{"commit":"25ed65f1fa1d50cf3d5f757736b6cfb506ce402f","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25ed65f1fa1d50cf3d5f757736b6cfb506ce402f"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-18 18:51:35.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-18 18:51:35.000000000","tz":-240},"subject":"Update os-brick library to 2.8.6","message":"Update os-brick library to 2.8.6\n\nThe previous version contained a regression that affected operation\nwhen running under Python 2.7 while using the Dell EMC VxFlex OS\n(ScaleIO) storage backend.\n\nChange-Id: I018007bd865b69631c21af83efdf6aba981161f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/00b1818979aba8e9abd5bef509a74fb27b5602af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/00b1818979aba8e9abd5bef509a74fb27b5602af"}]},"branch":"refs/heads/stable/stein"},"890875ee58aeb80cad1ea7268d680c78265359ab":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-24 12:25:49.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/736805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/736805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/736805/2"}}},"commit":{"parents":[{"commit":"25ed65f1fa1d50cf3d5f757736b6cfb506ce402f","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25ed65f1fa1d50cf3d5f757736b6cfb506ce402f"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-18 18:51:35.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-24 12:25:37.000000000","tz":0},"subject":"Update os-brick library to 2.8.6","message":"Update os-brick library to 2.8.6\n\nThe previous version contained a regression that affected operation\nwhen running under Python 2.7 while using the Dell EMC VxFlex OS\n(ScaleIO) storage backend.\n\nDepends-on: https://review.opendev.org/#/c/736784/\nChange-Id: I018007bd865b69631c21af83efdf6aba981161f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/890875ee58aeb80cad1ea7268d680c78265359ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/890875ee58aeb80cad1ea7268d680c78265359ab"}]},"branch":"refs/heads/stable/stein"},"6b1a67e861e3fe967250824df4e724887ada3b7c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-25 20:50:03.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/736805/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/736805/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/736805/3"}}},"commit":{"parents":[{"commit":"25ed65f1fa1d50cf3d5f757736b6cfb506ce402f","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25ed65f1fa1d50cf3d5f757736b6cfb506ce402f"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-18 18:51:35.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-25 20:49:58.000000000","tz":0},"subject":"Update os-brick library to 2.8.6","message":"Update os-brick library to 2.8.6\n\nThe previous version contained a regression that affected operation\nwhen running under Python 2.7 while using the Dell EMC VxFlex OS\n(ScaleIO) storage backend.\n\nChange-Id: I018007bd865b69631c21af83efdf6aba981161f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6b1a67e861e3fe967250824df4e724887ada3b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6b1a67e861e3fe967250824df4e724887ada3b7c"}]},"branch":"refs/heads/stable/stein"},"a14fd58f5dc1e4027e4148dfb1c40253ebfda8a3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-29 21:51:32.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/736805/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/736805/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/736805/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/736805/4"}}},"commit":{"parents":[{"commit":"6a158d22c4a1011cc86a725023704ba2d0f96cc7","subject":"Make greande jobs n-v for EM and oldest stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a158d22c4a1011cc86a725023704ba2d0f96cc7"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-18 18:51:35.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-29 21:51:32.000000000","tz":0},"subject":"Update os-brick library to 2.8.6","message":"Update os-brick library to 2.8.6\n\nThe previous version contained a regression that affected operation\nwhen running under Python 2.7 while using the Dell EMC VxFlex OS\n(ScaleIO) storage backend.\n\nChange-Id: I018007bd865b69631c21af83efdf6aba981161f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a14fd58f5dc1e4027e4148dfb1c40253ebfda8a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a14fd58f5dc1e4027e4148dfb1c40253ebfda8a3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
