)]}'
{"id":"openstack%2Fcinder~737707","triplet_id":"openstack%2Fcinder~master~Idcc949ed6adbaea0c2337fac83014998b81ff1f8","project":"openstack/cinder","branch":"master","topic":"bp/multiple-default-volume-types","hashtags":[],"change_id":"Idcc949ed6adbaea0c2337fac83014998b81ff1f8","subject":"Default type overrides","status":"MERGED","created":"2020-06-24 08:19:44.000000000","updated":"2020-09-17 09:09:50.000000000","submitted":"2020-09-16 21:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":265,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737707-1600292953614-c6eda660","meta_rev_id":"6c3018fb64ef26d854b6488178032c5bd9b202f8","_number":737707,"virtual_id_number":737707,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-16 14:42:32.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-17 09:09:50.000000000","post_submit":true,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 16:44:43.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"date":"2020-09-16 14:06:42.000000000","_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 14:57:04.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 21:49:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"date":"2020-09-17 02:04:49.000000000","post_submit":true,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-09-16 14:29:24.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 20:20:22.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-09-16 15:57:12.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 17:59:51.000000000","_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 18:36:04.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 20:30:01.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 18:50:20.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"date":"2020-09-16 14:10:20.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"date":"2020-09-16 21:29:39.000000000","_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-09-16 15:01:49.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 14:54:10.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 17:48:09.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2020-09-16 20:25:33.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 17:59:29.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 21:18:19.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 19:05:44.000000000","_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"date":"2020-09-17 03:41:12.000000000","post_submit":true,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-16 18:15:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":2,"date":"2020-09-16 15:38:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-16 15:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-16 18:15:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-16 14:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-03 13:23:55.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-07-14 15:56:22.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-07-21 16:02:56.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-12 21:08:53.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 20:51:32.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2020-08-31 04:30:19.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-09-01 08:56:54.000000000","updated_by":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"reviewer":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"state":"REVIEWER"},{"updated":"2020-09-08 17:38:39.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-09-15 11:40:34.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 02:46:11.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 02:57:38.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 03:19:13.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 03:30:26.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 10:45:54.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 12:28:04.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 14:06:42.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 14:29:24.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 14:42:32.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 14:54:10.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-09-16 14:57:04.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 15:01:49.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 15:38:43.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-09-16 15:56:58.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2020-09-16 15:57:12.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 16:44:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 17:48:09.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 17:59:29.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 17:59:51.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 18:01:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 18:15:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-09-16 18:36:04.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 18:50:20.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 19:05:44.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 20:20:22.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-09-16 20:25:33.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 20:30:01.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 21:18:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 21:29:39.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 21:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-17 02:04:49.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-09-17 03:41:12.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-17 09:09:50.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fc1267971b17b6f6e9d763d2ff27616f197ffec","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-24 08:19:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c18ed4feab667fedc03fc50b75ad3ffbaa38a9bc","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 08:22:28.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"446250496d918c1f9704a3d443ac268b49e80221","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-24 08:43:20.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/1/check-not-vote/ext-cinder-zuul/dc1bd34 : FAILURE in 22m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f56e3a3cc330452762495ec960a7f94ac4e995b1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-24 08:44:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/1/check/DellEMC_XtremIO_FC/d1fb36b/DellEMC_XtremIO_FC/245 : FAILURE in 24m 27s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/1/check/DellEMC_XtremIO_iSCSI/1e2b054/DellEMC_XtremIO_iSCSI/242 : FAILURE in 24m 09s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"399d44e0a38a8883f4d13218feeae9427124c661","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-24 08:52:21.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-1/ns5_nfs : FAILURE in 26m 22s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-1/ns5_iscsi : FAILURE in 32m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5d51eaf2d7b0448183d499e620e5dc06c73c1379","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-24 09:01:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a49d659/a49d659.html : FAILURE in 39m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8ccfc046a41b77a633129ea1df0f6cb616e7cd12","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-24 09:10:59.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3ae0b35373fe98e6fb8a57f00a71d9691541897f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 09:20:27.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c2123103326953b069c1ceb0a58eccb7cd4b8738","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-24 09:25:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/1/check/dsvm-tempest-cisco-zonemanager-job/dbb6891/ : FAILURE in 32m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ccc6c9f1a9e5da6226d7a91ae6579690ffda137e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-24 09:26:29.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/07/737707/1/check/dsvm-tempest-infinibox-fc/30c7764 : FAILURE in 40m 16s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/07/737707/1/check/dsvm-tempest-infinibox-iscsi/82255dc : FAILURE in 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4264cf9f0ff3b4a7a1d024d52ec97c12ec8c274b","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-24 09:50:26.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/dd08d7c2857907d2d55f62a0b98915084fb88756/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9354af24377ad1c06fef2b08f4207eca630f5359","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 10:02:39.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6b5bd7f106ba6176024d2c8bd3a3b6130e52872f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-24 10:05:46.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/1/check/tempest-dsvm-full-bionic/fdc6fb8/ : FAILURE in 51m 45s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/1/check/devstack-plugin-ceph-dsvm-bionic/2c7cf7c/ : FAILURE in 1h 04m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b5ec7a3989498974f1d67cabee2a0408d2622b18","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-06-24 10:18:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/07/737707/1/check/nimble-iscsi-driver-dsvm-volume/3d9f137 : SUCCESS in 1h 58m 15s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/07/737707/1/check/nimble-fc-driver-dsvm-volume/dd21f13 : SUCCESS in 1h 40m 01s","accounts_in_message":[],"_revision_number":1},{"id":"757bea93075ce662521625105fefbb65d87063f6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 10:39:55.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"380ac0ccb1f0f0837faf601947b2a00f7276ae6e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-24 11:32:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/1/iscsi/ : FAILURE in 3h 11m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/1/smb/ : FAILURE in 3h 04m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0190b2b758e8c997932fdb15dd0e6702a8dc39c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 11:48:59.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5b47c704b3e4e37a5eff298f645cb99 : FAILURE in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1bd41fd81744df0ae7af91706403747 : FAILURE in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e606a0dec2be4904b8cab161a4ed5728 : FAILURE in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c179b058430481e98084c4223335be8 : FAILURE in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/646337540bde4786b6602e2590ed16a5 : FAILURE in 6m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/9e322471adaf4f9c88eb6244daff6aa1 : FAILURE in 56m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1deefbac9f3840a88e08d42b95337ff9 : FAILURE in 28m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b56b32466bdc44a586e84b2d427d5867 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbed64d2d687416986f56fe4faca5820 : FAILURE in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec1008b2272342e2a6f52ec5631a8a89 : FAILURE in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5785555d1d554e5dbbb94c4fb9b80259 : FAILURE in 27m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3d38c33fa5f46508529c7b95dd3c453 : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0649fabcfc54aae8b84310d469575fd : FAILURE in 26m 47s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40e0123cbdcc498f850e4e7dc3644e1a : FAILURE in 28m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b2178490f99b40629b5f1e5f99f91790 : FAILURE in 1h 05m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c8de0879e31745358a4958edfd518458 : FAILURE in 24m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c5ed50a1a89740f0ac0c85e596bc45f0 : FAILURE in 27m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e3403f9cd9d4978808ae1f94eee39f2 : FAILURE in 28m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f93a145a47084034b682429d56b1e26a : FAILURE in 29m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dfd1a220fc750c80b88126d450e70462d8bfc90c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 12:08:17.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f4732b0003a9396d74fa598345c599743177680e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-24 13:44:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/1/check/DellEMC_Unity_FC/5516a83/DellEMC_Unity_FC/213 : FAILURE in 35m 00s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/1/check/DellEMC_Unity_iSCSI/1920007/DellEMC_Unity_iSCSI/206 : FAILURE in 24m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8d0661b33ad50481984881d67a3d53a98912a9ee","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-24 14:11:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/de557a8 : FAILURE in 20m 26s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/0961efe : FAILURE in 22m 33s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"8b6360ef2caf6ba5e5ad7a0118839425864b795e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-24 14:13:14.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/1/check/DellEMC_PowerMAX_FC/3a5a35c/DellEMC_PowerMAX_FC/138 : FAILURE in 25m 41s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/1/check/DellEMC_PowerMAX_iSCSI/440149b/DellEMC_PowerMAX_iSCSI/121 : FAILURE in 20m 55s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"9b7eaafc90a22778690df277416381fe99f3c74f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-24 14:43:04.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8fdb309a4ae6c2903a282b6ddda8af2197635593","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-06-26 15:02:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a1999e8ec553e4f8696f67d5f7ca9126c44aab8e","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-26 15:19:47.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/646f24e1419d340d286bf8441d39b2f4f893a69c/index.html : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0f14a47ce5d40778e0b3694779ed8beef94052a8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-26 15:45:26.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/2/check-not-vote/ext-cinder-zuul/87661f6 : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b7cd3918ad6be27383c87a4af423bce52d6155fa","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-26 15:53:20.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c76bb9a49f984ccb4cdccae75da6f409a84327e2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-06-26 15:54:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/5fd6924/5fd6924.html : FAILURE in 51m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c3d3f0273415d8f340d7d8dd9d5314b66a677819","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-26 16:59:53.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/07/737707/2/check/dsvm-tempest-infinibox-fc/14a5dc7 : FAILURE in 1h 09m 22s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/07/737707/2/check/dsvm-tempest-infinibox-iscsi/f9cd624 : FAILURE in 1h 06m 52s","accounts_in_message":[],"_revision_number":2},{"id":"17523a81996143c7d15ff0d40aea63a26f58344d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-26 17:08:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/2/iscsi/ : FAILURE in 1h 23m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/2/smb/ : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":2},{"id":"12850c66d9b61ddc9b1f51c3b65fb67541ad9ccb","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-06-26 17:31:11.000000000","message":"Patch Set 2:\n\n(7 comments)\n\nIf you are working on the client as well and it\u0027s not working you should check with --debug to see the URL you are actually trying to hit, as you may be going to the wrong location.\n\nMake sure you have increased the maximum version on the client as well or it will downgrade to 3.60 and then the server will return 404.","accounts_in_message":[],"_revision_number":2},{"id":"b472916d60f0c0b2e8eba1407ff77eee035499a9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-26 17:36:45.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-2/ns5_nfs : SUCCESS in 2h 34m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-2/ns5_iscsi : SUCCESS in 2h 27m 27s","accounts_in_message":[],"_revision_number":2},{"id":"00b89fc167ec364d8a786dcd9cd614df318cb779","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-26 17:47:22.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a9282a4b16a077e5cbb76572716dd87691d8d694","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-26 17:52:20.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/2/check/DellEMC_XtremIO_FC/e7ed412/DellEMC_XtremIO_FC/266 : SUCCESS in 1h 32m 36s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/2/check/DellEMC_XtremIO_iSCSI/857132b/DellEMC_XtremIO_iSCSI/263 : SUCCESS in 1h 35m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"2cba2fbb947de5600271068f2898fbda67fab134","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-06-26 18:33:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/07/737707/2/check/nimble-iscsi-driver-dsvm-volume/035ffb0 : SUCCESS in 1h 56m 07s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/07/737707/2/check/nimble-fc-driver-dsvm-volume/ba60c61 : FAILURE in 32m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b04b355021224437cbf6a66141506c1477183573","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-26 18:36:32.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"214098a1d85884008f7410572c4334c8d0a0ff63","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-26 21:52:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/2/check/DellEMC_PowerMAX_FC/3737c36/DellEMC_PowerMAX_FC/153 : SUCCESS in 3h 46m 03s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/2/check/DellEMC_PowerMAX_iSCSI/f73d5f3/DellEMC_PowerMAX_iSCSI/136 : FAILURE in 12m 59s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"31a975f28a70309e011e6aa8565ba9f76a02e8e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 21:55:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db0a4150d91742598114bd2be21b9fce : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bac14ea9974bdfa32caf53ff9e7b77 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f1ae91db6fb4a2c85b5877ce8a7e70a : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44bf9195f748469c9b590e35f497a0aa : SUCCESS in 16m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b075a5a7fe403cbdbad85bdc4d81f8 : FAILURE in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/e7284f33bc084e129a74c62b682accc5 : SUCCESS in 1h 07m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d6cb2349ccb4048ac00a8a5bc937f85 : SUCCESS in 1h 27m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ca3faedaf9374ad888645e810d7c2aef : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d62b03d9f6204b81b9b73c0073eef2a9 : FAILURE in 9m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45a1483026534584abe4f8230e90b137 : FAILURE in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89fbef9451054864b277678e096bfce4 : SUCCESS in 40m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f5791ca3a9c4c9b90b70ac88f26cd91 : FAILURE in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/961ef483256844df8d1f3d4a90e0a23c : SUCCESS in 1h 05m 48s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c99c6294d76f40f0a8813c005fb8f469 : SUCCESS in 1h 39m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/617fcafbbfc2435f9e6d45d6dbacb871 : SUCCESS in 1h 02m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9dee1f671bac4174bb750d51fc6ea319 : SUCCESS in 1h 21m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a258a392c7ad4d8e9951e01b3869d6da : SUCCESS in 1h 22m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af0c097497524cedbc70943e6bbd4e75 : TIMED_OUT in 3h 06m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cd4a5086da44947be72a578d43be428 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a095bb76badead27228445e524c238d5abddef57","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-26 23:25:40.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"102062330038251664d34cd5e785e0d7e5a02ea5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-26 23:34:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/2/check/dsvm-tempest-cisco-zonemanager-job/3961ba2/ : FAILURE in 33m 45s","accounts_in_message":[],"_revision_number":2},{"id":"fff0629cdea33af7b762e81303b41ca1b0ada783","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-27 01:28:15.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/d13f6cd5946e49fe8a49394bb4f51387 : POST_FAILURE in 18m 06s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/2/check/devstack-plugin-ceph-dsvm-bionic/2c3801b/ : SUCCESS in 1h 34m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ed4272da038f446ad58413c611ac33a5c1ebb344","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-27 02:03:01.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"03e650c5cd300448df08421fb5c91de077b6a316","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-27 07:43:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e83d7d4 : FAILURE in 20m 10s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/3dc5ff1 : FAILURE in 26m 29s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"2965918d022d07cafe57d3aab45e4a3832cd41b2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-29 06:19:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/2/check/cinder-storpool-tempest/ab295bb/ : RETRY_LIMIT in 11m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d62c72e62e7b60dbeb6222284d28d1c76d6f815f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 14:04:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/2/20200701203348/NEC-ISCSI : SUCCESS in 2h 15m 10s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/2/20200701203348/NEC-FC : SUCCESS in 2h 30m 31s","accounts_in_message":[],"_revision_number":2},{"id":"1ecd74171f6651de9d130f992c5ecefd47d0a173","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-07-03 12:09:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6005f7fca7f2b732894e2b231fc998ad5248235c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-07-03 12:11:47.000000000","message":"Patch Set 3:\n\n(7 comments)\n\nThanks Gorka for the review.","accounts_in_message":[],"_revision_number":3},{"id":"7c666f43ce03a7b7a2b2065b737ccb2eb0bfad35","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-03 12:39:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/dd14ce5/dd14ce5.html : POST_FAILURE in 7m 33s","accounts_in_message":[],"_revision_number":3},{"id":"c3166af86f539ec5c8e0054f2235bf6434eaf5c5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-03 12:42:51.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b38acac58c994677e1a7ba0380c65f68f7069354","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-07-03 12:52:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/3/check/cinder-storpool-tempest/a255290/ : RETRY_LIMIT in 11m 36s","accounts_in_message":[],"_revision_number":3},{"id":"b23040a26c8a3eb120efeee248a3c3f9a90f9dc7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-03 12:53:14.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/3/check-not-vote/ext-cinder-zuul/af68326 : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b7cf10a605b8dbd65e963a619220119fc5c11c9e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-03 13:00:57.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"e44000797ea32b15dc624226d3426b686759d979","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-03 13:08:28.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/3/iscsi/ : SUCCESS in 58m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/3/smb/ : FAILURE in 50m 21s","accounts_in_message":[],"_revision_number":3},{"id":"88584f535a45532132cdd434dd28adc03530a596","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-03 13:21:51.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/463b132 : FAILURE in 20m 57s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/8bbc341 : FAILURE in 20m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"7aa552e5cf28f1e80761e1fac9a24a65befa78d1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-07-03 13:23:55.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/07/737707/3/check/dsvm-tempest-infinibox-fc/4343ae7 : FAILURE in 1h 14m 22s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/07/737707/3/check/dsvm-tempest-infinibox-iscsi/313a2c7 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"304acddb61e661ae2c41bb2f8936e2be752e8641","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-03 13:29:01.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/7466b565af09c34b3d3afa528774706875c00928/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"1252f2170e6f68a59ea1041f31f0ce509de18e86","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-03 13:33:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/3/check/dsvm-tempest-cisco-zonemanager-job/c297724/ : FAILURE in 33m 16s","accounts_in_message":[],"_revision_number":3},{"id":"de2a953d9c9ebf153eb4f7c554435f70b860bcdd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-03 13:43:40.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/3/check/DellEMC_Unity_FC/e50ab42/DellEMC_Unity_FC/260 : SUCCESS in 1h 17m 19s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/3/check/DellEMC_Unity_iSCSI/7af3398/DellEMC_Unity_iSCSI/250 : SUCCESS in 1h 34m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"14aba9a7e02154ef31066cc6f338c6398cbc3609","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-03 13:44:06.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2f3996c6e60644308339dd4e610dcedfb0c8485a","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-03 14:09:01.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/07/737707/3/check/nimble-iscsi-driver-dsvm-volume/ba85e11 : SUCCESS in 1h 59m 40s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/07/737707/3/check/nimble-fc-driver-dsvm-volume/98849b1 : FAILURE in 40m 36s","accounts_in_message":[],"_revision_number":3},{"id":"209d72e2ef7d9f334b0972c04cf13f3c8d30046b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-03 14:38:11.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-3/ns5_nfs : SUCCESS in 2h 27m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-3/ns5_iscsi : SUCCESS in 2h 28m 43s","accounts_in_message":[],"_revision_number":3},{"id":"8cf52e0c1c160c03271bb2a1bd581cc7746c481c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 15:38:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fafc36713f514938bc68b77e249150ec : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a41e3a8ed14e3faf27e30b37512d53 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eb767e97d364c7c90d73096c09f0b35 : SUCCESS in 12m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdb956a034e44ce4820f0c7313214805 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e6f2014a4e40f3839b2d273996f244 : FAILURE in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/c19770630dde402a8b7fe2072a8d250b : SUCCESS in 1h 02m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea2e3f22be704e9fa0c41e7db5c88646 : SUCCESS in 1h 24m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3aaaa9e7befa40a28bbd3fc7bc54a1e3 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/57f1f2b212754cbab966dca953b080ae : FAILURE in 7m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c2f54bd8f8434b548c42f1c4f5b4efd6 : FAILURE in 9m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d357dbf3eaca4727adf38c8077ed27db : SUCCESS in 38m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/406bc93cc3714300af569befc58882ff : FAILURE in 8m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d736262118e04d03b12ffba2113cfe0e : SUCCESS in 1h 25m 57s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/75c13e2e68904bf585c96fb3d64ff987 : SUCCESS in 1h 53m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7f69d704cdf543bca83af92ba874f042 : SUCCESS in 1h 34m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e638346fcfc6414c975a8f6f20503438 : SUCCESS in 1h 12m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bc4b51b5a1ae49b7b3289673c7521656 : SUCCESS in 1h 28m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e646dfb7bbe1432e969f50b3f4222ce5 : SUCCESS in 1h 57m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae13d49a41234ee1b23e68d4f59be4a9 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":3},{"id":"eed8d929c7eefc97f38c94be023a5418433e6615","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-03 16:11:21.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/3/check/tempest-dsvm-full-bionic/d34d0d5/ : SUCCESS in 1h 41m 42s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/3/check/devstack-plugin-ceph-dsvm-bionic/c21a861/ : SUCCESS in 1h 40m 07s","accounts_in_message":[],"_revision_number":3},{"id":"ab95a02bf991be19acf46f6d404af8dbd0f22ef6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-03 16:47:28.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9def1139301ab135d01788b8f8cee566e619b7c4","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-07-03 19:53:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1288/index.html : FAILURE in 26m 23s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1435/index.html : FAILURE in 30m 24s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1460/index.html : FAILURE in 26m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f204a77b937b0b842ff95044fc876606bc0c7cf8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-06 12:25:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/3/20200706191059/NEC-FC : SUCCESS in 2h 07m 56s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/3/20200706191059/NEC-ISCSI : SUCCESS in 2h 14m 08s","accounts_in_message":[],"_revision_number":3},{"id":"45f6cc44282757a434935fb30689d28a1b519e22","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-07-09 06:55:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b594cc058e679543641f495746576cc15c4b1ab8","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-09 07:03:30.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/09f2139557a2ed29c15c846f7191fc49bb6eadf9/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"641cc2a081df24f4db631fb2dd675b4533341b34","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-09 07:10:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c310545/c310545.html : POST_FAILURE in 7m 39s","accounts_in_message":[],"_revision_number":4},{"id":"fe188087dc408190399a1b463f7e9611ba751ffe","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-09 07:10:20.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1702 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1909 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1737 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a9ad3d8f59dc18abda695ecc24e64d96b134f8ad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-09 07:34:19.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/4/check-not-vote/ext-cinder-zuul/f834938 : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":4},{"id":"2a776d35f2477e38e2f74c714a76aadcca93db3e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-09 07:48:09.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a8d777549ba3e56bd8733364536556ce06b3c21b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-09 07:56:05.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/4/iscsi/ : SUCCESS in 1h 00m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/4/smb/ : FAILURE in 53m 41s","accounts_in_message":[],"_revision_number":4},{"id":"b0e94304fa66c5c5eb3c998a2577164b92e02484","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-09 08:21:09.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"d7936967666fa3f029bed176e8d6b623da9f370b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-09 09:26:45.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/4/check/dsvm-tempest-cisco-zonemanager-job/6befc44/ : FAILURE in 34m 23s","accounts_in_message":[],"_revision_number":4},{"id":"74effac8f6181ecb3b62722af0fc694b1cc72f7b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-09 09:35:55.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"b70ba13d7222f6abefbb7f788487bdd2d3e9c213","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-07-09 09:59:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/4/check/cinder-storpool-tempest/38ec76a/ : RETRY_LIMIT in 11m 59s","accounts_in_message":[],"_revision_number":4},{"id":"8d6b673b8a50e92358fecc6c8666589d1152dbed","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-09 10:16:43.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b99e582 : FAILURE in 20m 27s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/7fc34e7 : FAILURE in 20m 38s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"1e51a9f3c376f1eff8346270446d5bc7a01078d1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-09 10:28:01.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/4/check/tempest-dsvm-full-bionic/b4a8088/ : SUCCESS in 1h 29m 15s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/4/check/devstack-plugin-ceph-dsvm-bionic/1a890c2/ : SUCCESS in 1h 48m 52s","accounts_in_message":[],"_revision_number":4},{"id":"52ffc85b053603951e1ecba0ca7301245b66973e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-09 10:57:33.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-4/ns5_nfs : SUCCESS in 4h 1m 43s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-4/ns5_iscsi : SUCCESS in 3h 45m 33s","accounts_in_message":[],"_revision_number":4},{"id":"6f07627f286340b96e52aeeb3324e67a5c5c536a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-09 11:55:37.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"f7f4ce54c8efe6d66750b0291d399f39c6563650","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-09 12:02:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/4/check/dsvm-tempest-zadara-driver/3b7f77b : FAILURE in 1h 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7bbf9c2e0e0cadf5b3c4a800c6f2b1894092c156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 14:19:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5f6cde131aa4137a67fcf091ba571bb : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ce13c759c14b519816834b4f666bc1 : SUCCESS in 9m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88b41b0ec1324b95a70fa9b9c6e2c749 : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a336e7f5e24c457b82b3b64e46d1c977 : SUCCESS in 11m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/109ef08ff1644170ad31996ced88c70b : FAILURE in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/3c45fa5ee173445cbc654cdeb0200e01 : SUCCESS in 43m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee9a6781be734c798406b45106e04594 : FAILURE in 1h 16m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/89ad908894fb44fa9948e5ac832da59e : SUCCESS in 17m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8e8eb0830a74597adcbee697d8f960e : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e72593cf09948dfb6b14bd50e0e9ec3 : FAILURE in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f3401f7439f4dd08f5a6865240c2d69 : FAILURE in 8m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0f9e6b2679d04748907aa8414a81727b : SUCCESS in 41m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4afe671d37864db7b65856b541a9f5be : SUCCESS in 10m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/386a01abeb4e4f428017715b915ac0ad : SUCCESS in 1h 09m 34s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a570ce97611944e08876fae0c1589422 : SUCCESS in 58m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8cc31253875a4bbea2fb195cd3181a5f : SUCCESS in 51m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/675109f9630349e6a3bd9698e2ec4b8d : SUCCESS in 1h 32m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/72000ac35be34cb5be0145f9c5f39272 : RETRY_LIMIT in 2h 15m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa12c533d9224efe91ef7b42813ae541 : FAILURE in 2h 44m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5debe49ed95433a933f3a5dcef98fac : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":4},{"id":"80e7044594ae53030d4bd40acb2b8819f4b04e02","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-09 14:40:16.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/4/check/DellEMC_XtremIO_FC/771a34e/DellEMC_XtremIO_FC/331 : SUCCESS in 1h 31m 52s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/4/check/DellEMC_XtremIO_iSCSI/91b4561/DellEMC_XtremIO_iSCSI/321 : SUCCESS in 1h 34m 18s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"ce40e593ba1b5922382122a34da4b1404a965123","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-09 16:49:39.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"465629291f9eb814534eeeb1a49894280efff04f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-09 17:30:06.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/07/737707/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"4264c1f5087a9bce5dbdf1430f27dc6313dc67f3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-09 18:42:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/4/20200710012554/NEC-FC : SUCCESS in 2h 05m 42s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/4/20200710012554/NEC-ISCSI : SUCCESS in 2h 16m 10s","accounts_in_message":[],"_revision_number":4},{"id":"4cae25ee8b84f8f0752d4ba52770602d1834c84c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-07-14 15:30:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1e752d840307f281257fa4a1248c6ec2f113504f","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-14 15:35:36.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/267192337a46ffdbf29eccf136c90c3dbca6a499/index.html : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"1894e0a340d72ca0d4a971ed8274eeee6ada2b35","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-14 15:52:58.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/5/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/d212f5a : FAILURE in 20m 49s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/5/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/07dc8a1 : FAILURE in 20m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"0f72ff472cc64f8a264a6062fba942ef99170933","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-07-14 15:56:22.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1726 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1897 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1691 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"85e8877d772acf4b00802e011bb6fe6bc757528c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-14 16:07:27.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/5/check-not-vote/ext-cinder-zuul/74cf9b0 : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":5},{"id":"a5b164e319756cb707dc424d57cd6183b6524706","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-14 16:21:34.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"f51add52c3530cf000847972a1ac81eac6151c74","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-14 16:28:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/5/check/dsvm-tempest-cisco-zonemanager-job/4b862ea/ : FAILURE in 32m 59s","accounts_in_message":[],"_revision_number":5},{"id":"1b160d32b8e03fdbb9de9e696f01bc409745d7d7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-14 16:34:49.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"d455cec38d50f375ab957a1b781e1e795e68adba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-14 16:40:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/dcde2a1/dcde2a1.html : SUCCESS in 1h 08m 27s","accounts_in_message":[],"_revision_number":5},{"id":"eba1fa2166809df6f18964eff1f55e173035f0f4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-14 16:47:08.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/5/iscsi/ : SUCCESS in 1h 07m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/5/smb/ : FAILURE in 1h 16m 21s","accounts_in_message":[],"_revision_number":5},{"id":"33df43c697ad9227e0e84dc06826db03187e7cd4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-14 16:48:18.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/07/737707/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"5833bcc84ad94b00b4224e9f576102c52efc287f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-07-14 16:50:24.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/5/check/DellEMC_PowerMAX_FC/ae15faf/DellEMC_PowerMAX_FC/269 : ABORTED in 51s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/5/check/DellEMC_PowerMAX_iSCSI/358ec33/DellEMC_PowerMAX_iSCSI/213 : ABORTED in 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"e98c783c0f9d9caac11ace48e6c98b6464c576fd","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-14 17:17:40.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/5/check/DellEMC_XtremIO_FC/d210ff1/DellEMC_XtremIO_FC/385 : SUCCESS in 1h 34m 17s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/5/check/DellEMC_XtremIO_iSCSI/bf9b773/DellEMC_XtremIO_iSCSI/357 : SUCCESS in 1h 35m 55s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"f412e868320cc82ec3dc33bdf2aaab0bfddd31f3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-14 17:41:51.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-5/ns5_nfs : SUCCESS in 2h 10m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-5/ns5_iscsi : SUCCESS in 2h 11m 15s","accounts_in_message":[],"_revision_number":5},{"id":"fc86f2dce9d7840517ede5d6e30edb5626668c45","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-14 19:08:41.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"d401930337190600896017110dbba7a9c3538108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 19:11:57.000000000","message":"Patch Set 5: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26bf8b1bf9bd46148a02ead36a533a0c : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7438a7b8c77944e88840e12a98974f7d : FAILURE in 16m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e058d40bab184abc81c7ce3821e52a86 : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e30909c368e04779a75d6bd182b306c7 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc34a8d3a9794e7d98042390518346c0 : FAILURE in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/e2453ddd164a427eae33b40ccc45f5a8 : SUCCESS in 1h 11m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3006d53092294e76a3b943c511a4dd70 : SUCCESS in 1h 16m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a34a9a247c1e4d0b894f661c02b65d95 : SUCCESS in 11m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb1142413aee4888907c29c0ec994106 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5272692db74d41729a57295b027ece2e : FAILURE in 8m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1fe51d4cb9614232a42f05932de48fc6 : FAILURE in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/185a39f7c06f433d98e6bf0a849e3770 : SUCCESS in 45m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bef0cfd98e3b473989ae8331d2a4f116 : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b1783ceac05f4fcfae7f9aa57abae56c : SUCCESS in 1h 05m 15s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/482205eb10c04db7be85012d882f5a18 : SUCCESS in 1h 29m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/31846f2bfeed448ab76fed8025215a79 : SUCCESS in 48m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/36a5b3c0cdca40c5a41ec7c1ca73d74c : SUCCESS in 1h 16m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/42c749bed483407680ff61bfdf12946c : TIMED_OUT in 3h 11m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5298ae9b3075431785bc6789cc7b1a39 : FAILURE in 2h 05m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf809e61f7374253883049e73f313807 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":5},{"id":"60fef8cc7220125cbb533f1154b72eaad04df2de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-14 19:27:00.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/07/737707/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"97ee38aa42c8aef23275dadc7206df1ad5b77aba","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-14 19:41:02.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/5/20200715022151/NEC-FC : SUCCESS in 2h 09m 58s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/5/20200715022151/NEC-ISCSI : FAILURE in 2h 19m 08s","accounts_in_message":[],"_revision_number":5},{"id":"238cdb5721813c784e2b20991d56f1636609659d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-14 21:56:41.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/07/737707/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"453297b00575cf4c8ad36b0be10f03fc01720408","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-14 22:58:23.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"57e4f9e930e3f262fbbd2fedcfe6878cf7b1b014","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-15 13:59:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/07/737707/5/check/nimble-iscsi-driver-dsvm-volume/861763d : SUCCESS in 2h 24m 37s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/07/737707/5/check/nimble-fc-driver-dsvm-volume/f967b66 : SUCCESS in 1h 51m 10s","accounts_in_message":[],"_revision_number":5},{"id":"07f787e9390ecf79229eeb36851aa248c714529e","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-07-17 15:29:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1340/index.html : FAILURE in 31m 47s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1486/index.html : SUCCESS in 1h 37m 36s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1511/index.html : FAILURE in 1h 19m 09s","accounts_in_message":[],"_revision_number":5},{"id":"bd85f110689d129b8601d4549a037c1815dc40f4","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-19 16:12:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/5/check/dsvm-tempest-zadara-driver/28b75e0 : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c2f0403764697eab98b2d0a115930dd42456a001","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-07-21 12:42:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f5c0246858bdfae262d798d6ad493b8c396c4307","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-21 13:52:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c5059a4/c5059a4.html : FAILURE in 52m 17s","accounts_in_message":[],"_revision_number":6},{"id":"78da6f203037f183eeb2da1f72ff04c58a13d478","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-21 14:05:47.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"e42e86600829bb94111775ed52168a9599cf8247","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-21 14:08:25.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/6/iscsi/ : FAILURE in 1h 19m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/6/smb/ : FAILURE in 54m 11s","accounts_in_message":[],"_revision_number":6},{"id":"1d29c20fc560435042fea210c55226e3fabe7377","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-21 14:44:44.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-6/ns5_nfs : SUCCESS in 1h 58m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-6/ns5_iscsi : SUCCESS in 2h 1m 57s","accounts_in_message":[],"_revision_number":6},{"id":"1add66625a5806be311422b5e940acfc76279277","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-21 14:54:38.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f8ecda5 : FAILURE in 20m 26s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/e2c4140 : FAILURE in 27m 28s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"c8747de7f784cf2442917d3cf8925d279e6d4567","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-21 15:19:27.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"d22879992ecf11d707584dde4ffbdf46a9065e25","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-21 15:51:24.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/6/check/dsvm-tempest-cisco-zonemanager-job/7fb7501/ : FAILURE in 32m 44s","accounts_in_message":[],"_revision_number":6},{"id":"4c09c727d7db303df3280586810eb86643b0df80","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-21 16:02:56.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1721 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1928 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1754 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"e3f304f8b6726179324681975abb25546c7746c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 17:51:30.000000000","message":"Patch Set 6: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d6225305c2f42c3a47391bd27d31a5c : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ca135d7c8244fcbff8741796962d79 : FAILURE in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d95d9ab4c6234689a71de6ade39c8a72 : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c09ddf8275244afb80c9c24d602fa647 : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b5aca2d75c3467680b1c23c7096ad49 : FAILURE in 7m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/500f730d0a2b4403a3026c392a723ee7 : SUCCESS in 1h 02m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c05f0b309c946549b1b462d37dc7a73 : SUCCESS in 1h 20m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/944134630df04e0eaab8b5947cd9cd09 : SUCCESS in 12m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/07cb7d8977fa4a7c8d095109953d6a48 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a332d05673f4679b10e5f1b87f2d66c : FAILURE in 6m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/da703cd04a9c4de796f9fe36f4e182fa : FAILURE in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/da4d43a696b54a92893073433d28030f : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9981a9de9e9a42e2b918100f24ea6b05 : SUCCESS in 7m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/773a76dcfa4c4192b159352d33aa15ef : SUCCESS in 1h 15m 30s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66d80d1b03b24bde9fddba0d16e00140 : SUCCESS in 1h 46m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e7c4aba69ca4ee98ad8dc3bd09e6f78 : SUCCESS in 1h 20m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1da9745383b1470f8676b68ad9074730 : SUCCESS in 1h 21m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/199b68f8e1844d5f8a20da7f1bad7f4c : FAILURE in 1h 03m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0419ec3f71f14fcca66187e4eac87b15 : SUCCESS in 2h 15m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee10651eae404911a288b365841a8d3a : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":6},{"id":"2db60b9659293245588be42a9c47fd838f258f64","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-21 18:50:40.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/6/check/DellEMC_XtremIO_FC/5a07c76/DellEMC_XtremIO_FC/454 : SUCCESS in 1h 31m 21s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/6/check/DellEMC_XtremIO_iSCSI/e07b123/DellEMC_XtremIO_iSCSI/426 : SUCCESS in 1h 34m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"1560be3f1a3bf10285ff2cc497e496ad8f0169d8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-21 18:52:08.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"87c0075f2586ddbf2ae8b741f114c998ccff3dee","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 20:15:49.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/6/check-not-vote/ext-cinder-zuul/5ac36f9 : SUCCESS in 41m 42s","accounts_in_message":[],"_revision_number":6},{"id":"f96bb10fe626f7813d526cb1daa36e8da8f536b6","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-21 23:04:28.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/6/check/nimble-iscsi-driver-dsvm-volume/c205b8a : FAILURE in 27m 25s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/6/check/nimble-fc-driver-dsvm-volume/c141c09 : FAILURE in 56m 16s","accounts_in_message":[],"_revision_number":6},{"id":"b6c3deeafbee8ac21c97c9396f6318e02892efee","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-22 01:42:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/6/check/dsvm-tempest-zadara-driver/fd8d96a : FAILURE in 1h 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1150475e58270fa27d274ee7eadd0c4f4caeace2","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-22 04:00:10.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/4c69444336f15ab92a342a9b2e853211203eea96/index.html : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"e8ddfebb25787f8b8b3617d9f279b67176bed2f5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-22 14:10:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/6/20200722203112/NEC-ISCSI : SUCCESS in 2h 22m 12s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/6/20200722203112/NEC-FC : SUCCESS in 2h 39m 29s","accounts_in_message":[],"_revision_number":6},{"id":"3e492bae6d09ba3b480bb7bad28b7715bd60fd3c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-12 15:38:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4a2acd85f0880b32c4e1ff05f84aedf2ea8494a4","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-12 15:42:57.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/5c328aa : FAILURE in 4m 19s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b0a78c6 : FAILURE in 4m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"7a3fec2fefb7052b4402580c0d506383bad87c76","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-12 15:43:53.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/acfaa9ede3386584520e80665b985ee055c7cbd9/index.html : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"23e71936f848525d8f4dd103dbb5994fbf02c256","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-12 16:02:28.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"8ab76e7d288d42f1f0e128f46f77ce4937f0fa20","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-12 16:33:21.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/6c472f3/6c472f3.html : FAILURE in 52m 56s","accounts_in_message":[],"_revision_number":7},{"id":"c5b67718b941990143c4d6f95c6e2e6410f556f2","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-12 16:45:18.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"52bac68c0c41d91eb91ea8ccd28080f0716a6f3d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-12 16:54:47.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/07/737707/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"46ccfaff740fb23eda8d486dc9e6ec441e653978","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-12 17:06:27.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/7/iscsi/ : FAILURE in 1h 27m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/7/smb/ : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":7},{"id":"063de59fa2d91796ff90c5f823c518b77851f978","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-12 17:29:41.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/7/check-not-vote/ext-cinder-zuul/8698a45 : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":7},{"id":"38686a0258b9b1b97eaef9792348a7c2f56087c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-12 17:38:30.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/7/check/tempest-dsvm-full-bionic/dd446c8/ : SUCCESS in 1h 33m 05s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/7/check/devstack-plugin-ceph-dsvm-bionic/41c43f1/ : SUCCESS in 1h 37m 44s","accounts_in_message":[],"_revision_number":7},{"id":"f874bc3e238929e737741688ed3f30bfe6926c5f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-12 17:46:05.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-7/ns5_nfs : SUCCESS in 2h 4m 15s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-7/ns5_iscsi : SUCCESS in 2h 7m 26s","accounts_in_message":[],"_revision_number":7},{"id":"bf5dd922b621ce27c9275999f46f2195969a2ff6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-12 18:03:43.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/7/check/dsvm-tempest-cisco-zonemanager-job/648034e/ : FAILURE in 32m 54s","accounts_in_message":[],"_revision_number":7},{"id":"2a1e886c1014bf71af001f15ad2e0f1c89829a59","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-12 18:03:54.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"17485d2f41aebd493fb6c7ddb3bafd045b13c0e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 18:18:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e365d2c6fdc4fc3ba102459add32290 : FAILURE in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd9992b5dc84cd48f4f980a62c57056 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/029bbdcbb3174422b73b5d06a2af178d : FAILURE in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f0115867db24dfebd4ee3f768566396 : FAILURE in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4c4dc521d944ec877adb59bc8bc049 : FAILURE in 7m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/e70c1f6b1cf14facb511e8a6ec1e6472 : SUCCESS in 1h 03m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a22aa32720f54dc19f29f329a473ad15 : SUCCESS in 1h 22m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95db0b7063b24cc0bf15b5a715c50058 : FAILURE in 9m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8edb1ccc54bb4269985e0cd1069c66a8 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ef362688fe2472cb0fdd6e4f441a887 : FAILURE in 9m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cffe6dccd093487ca9abe1a3d9230968 : FAILURE in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/58cfdc53ba964fa8925f269fe8d2d406 : SUCCESS in 41m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ac904d63d06401d87a180738ddff58a : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3c64ba04d1764fd3ad059c127f11fdc9 : SUCCESS in 1h 25m 28s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bee48576579a49b5bf35b6e37021e97c : FAILURE in 1h 44m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8d230b62787340f4ab7dc7e81b5ce5da : FAILURE in 1h 07m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/13b3c27460584cb7a735b4c96676757f : SUCCESS in 1h 20m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bbe10a5d8b68408580b3aa9e8dfb992d : SUCCESS in 1h 17m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c41671186ca2429482ec1f9c8078121c : SUCCESS in 2h 24m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/373cafe3d18542029aa55854dfa31e59 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":7},{"id":"2b3b2ff29808ab360e072433b4d895a964f7b8a8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-12 20:42:34.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/7/check/DellEMC_XtremIO_FC/04588b4/DellEMC_XtremIO_FC/663 : SUCCESS in 1h 35m 04s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/7/check/DellEMC_XtremIO_iSCSI/7c5467f/DellEMC_XtremIO_iSCSI/634 : SUCCESS in 1h 37m 41s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"49dcefa2f2920f6944ddcc4b024ec56eaa58776a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-12 20:52:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/7/check/dsvm-tempest-zadara-driver/75564c3 : SUCCESS in 1h 05m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"599c2c500b76a1fb714e808da9d8b3a2b790f4fc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-12 21:08:53.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/07/737707/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"3652ed903819e0baea02a44db3e03903193b77a8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-12 21:19:24.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"cde940f1e4e29f43e13a04048819c4c918eb61c9","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-13 06:11:19.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/7/check/nimble-iscsi-driver-dsvm-volume/144dd33 : ABORTED in 42m 17s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/7/check/nimble-fc-driver-dsvm-volume/d6dada6 : SUCCESS in 1h 56m 45s","accounts_in_message":[],"_revision_number":7},{"id":"9bde39a1682eacf754555efa1c3a739c0fb3853a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-13 11:39:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6f61053d7a64d018dda1d1c52443f4ef2c042b7f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-13 11:45:14.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f25f09e : FAILURE in 4m 56s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/4a778a1 : FAILURE in 4m 23s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"61f0575fdd4e0252e3428766805e8adf773068a7","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-13 12:31:32.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"94acd8eb3391ac72397838d62c5e63f19f52c7a1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-13 12:35:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/beff7e6/beff7e6.html : FAILURE in 52m 15s","accounts_in_message":[],"_revision_number":8},{"id":"0045b2a13f2aebce1239a32ea2ff187c860adaab","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-13 12:50:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/8/iscsi/ : SUCCESS in 1h 03m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/8/smb/ : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":8},{"id":"47270ac995f3abe90ac88caee79e3a9a3fcf816e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-13 13:14:03.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/8/check-not-vote/ext-cinder-zuul/d93dc6d : SUCCESS in 39m 42s","accounts_in_message":[],"_revision_number":8},{"id":"6a04adfbee2dca52545cb5df309462b57fbc0002","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-13 13:37:51.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/1295878d746698011e295a9f7ff30ae880bf98bf/index.html : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"1781a9275e8fcaaabe3734c8db1d4080f8404b87","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-13 13:43:09.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-8/ns5_nfs : SUCCESS in 2h 2m 51s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-8/ns5_iscsi : SUCCESS in 1h 49m 7s","accounts_in_message":[],"_revision_number":8},{"id":"65d5e776fdf101d37fd383d6166293943bad3ac7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 13:53:52.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"a4a6af3f475b80e2b447ebc810d3b6e35e1699cf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-13 14:21:50.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/8/check/tempest-dsvm-full-bionic/ff35d65/ : SUCCESS in 1h 37m 50s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/8/check/devstack-plugin-ceph-dsvm-bionic/513cf2f/ : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":8},{"id":"0b6930a9db7b20ed6ac7c1d4a19fe02aee4214b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 14:52:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ee4bf3658c334677be21a2dd6b9ea2ae : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da2d3a657f4145238b7a90d9a545646f : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7d9b6e5f8847d28928b78bb1008a7a : SUCCESS in 9m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fed2b5832d0f41509181615ffab3899d : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a2877fb059e4c2296bc2daf02ed1f23 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67eeadfa3d2b4b44aef795699650ea3b : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/8f9be4ed34ca4ca09aefea6b35c3b722 : SUCCESS in 56m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8098458f311c42729c7d218ab847a3ea : SUCCESS in 1h 16m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/930519b3c7ce4db2b817d120c2f29e54 : SUCCESS in 12m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ee9343ac84543c9bf8c9eb892b70059 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/276615635dc74f8e90b8f6704762145b : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dee05a990595483cb9cf57277445f295 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4b4a75688d414c7c973cc916b2e32ecb : SUCCESS in 40m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/245bc5e104174ddfa1e12d069533c53f : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47f52959ec744af090f3791499b2c581 : SUCCESS in 1h 11m 18s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f225bb79785425f868b4f419c1ffafa : FAILURE in 1h 29m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d1b71043e3af434d86fb1e0712e2ade7 : SUCCESS in 59m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f91f8c49f0b24dd0bdbd9ca41cd9de8a : SUCCESS in 1h 00m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/693e3e560d5e46a3bd564c97ae275b0e : SUCCESS in 1h 15m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/107f3017e97b4eb481cce5d346675e0a : SUCCESS in 2h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/547dd2e478d049278237800284235522 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":8},{"id":"3e97abd8fe737c7bd485e0a2e11428b96f2bac00","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-13 15:52:40.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/8/check/DellEMC_XtremIO_FC/c8ff7dc/DellEMC_XtremIO_FC/669 : SUCCESS in 1h 37m 48s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/8/check/DellEMC_XtremIO_iSCSI/527ae99/DellEMC_XtremIO_iSCSI/640 : SUCCESS in 1h 37m 10s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"7110490555382522d091f4b664cb2c93db785d24","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 16:27:00.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"d78104f7489c2bf73e274373fd810012fed40fa8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-13 16:51:34.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/8/check/dsvm-tempest-cisco-zonemanager-job/b777d8c/ : FAILURE in 34m 25s","accounts_in_message":[],"_revision_number":8},{"id":"a49233dd8c69a897684e07a149d76788a1b2f217","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-13 17:16:56.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"8f0898f8607da3a61f1236e1d0d09a88bddd7e3a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-13 17:25:16.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b166d44 : FAILURE in 4m 26s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b5a4dd3 : FAILURE in 4m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"b44cb85d890709012c874577f6f930c8aa1398ac","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-13 18:07:58.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"45067ccbba830642d198208e880747e165e53775","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-13 18:21:53.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/8/check/nimble-iscsi-driver-dsvm-volume/e3dca55 : FAILURE in 2h 43m 24s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/8/check/nimble-fc-driver-dsvm-volume/814552d : SUCCESS in 2h 31m 39s","accounts_in_message":[],"_revision_number":8},{"id":"32a674a92c1b93c56a2890761c6aece5738bfd18","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-13 18:28:21.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/8/check-not-vote/ext-cinder-zuul/79b2d48 : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":8},{"id":"5547d4a5ffb18b205962e4bcf1531f02ed5ebcb9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-13 18:39:19.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/8/iscsi/ : FAILURE in 1h 20m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/8/smb/ : FAILURE in 1h 03m 16s","accounts_in_message":[],"_revision_number":8},{"id":"19a5f4ee5ea9392b98eb64ee4030125f1609b78b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 19:01:08.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"1c7040b0995fe2db37d3451188a263630176a6f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 19:29:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/07a832ac1fe34644a61f6ae416f055bb : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6a2ccf2d49f4019bb3ce450af3cef6a : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c86c4d10daff4cc7b83af42138c0ec31 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b20cb7b4514c01a0721825ceb796ee : SUCCESS in 13m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f35bcc681df04dd79a47337141c78e31 : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d993dcd67a884ddeb99cddef8f6f183d : SUCCESS in 14m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/c167bc0885f34f86b175e2bc977c25cf : SUCCESS in 56m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dbe2b5a1e72247d9940e6367e4c8c5ef : SUCCESS in 1h 01m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ac9f2690aab4567af40b6c13c056e02 : SUCCESS in 11m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/080378e76aff46a894adab506b579b03 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c3a5140a9004321aa41a2bc1c7c9332 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/91401432ea634c45b6cd74ec81ad8e3b : SUCCESS in 7m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ace4380702db4e35bd94d5dd72872d1c : SUCCESS in 49m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10aacc93a97844ceb062e9699353ca62 : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c0ed8014d5d9475d9cb40f062f73e5e5 : SUCCESS in 1h 18m 21s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/38402e45f25249388ccae25bd33b51e8 : SUCCESS in 1h 26m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/84f7e41832744238bbad670674ef7438 : SUCCESS in 1h 15m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/899901fe4afb45e08c6df7530db21b1f : SUCCESS in 1h 14m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7d3f8507167945dab0449fe08434fd4f : SUCCESS in 1h 13m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c0a12078bc5411bacd39b069cf98009 : SUCCESS in 1h 59m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e39da349da044faea05b4fa75f3542bd : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":8},{"id":"d7c17156675d2243674dfca3f8600da458bc118f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-13 20:05:57.000000000","message":"Patch Set 8:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"81a3b710f6906f57b9aa59b303b94ec384e2dcee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-13 20:28:06.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/8/check/tempest-dsvm-full-bionic/8e32ac0/ : SUCCESS in 1h 32m 44s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/8/check/devstack-plugin-ceph-dsvm-bionic/357c31e/ : SUCCESS in 1h 47m 10s","accounts_in_message":[],"_revision_number":8},{"id":"7656a78acffa93c8a65520535b8618012984188d","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-08-13 20:40:07.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/8/check/synology-dsm-cinder-iscsi/c279f25 : FAILURE in 27m 22s","accounts_in_message":[],"_revision_number":8},{"id":"28a9ca7030fba47a0b6e0320b655f887adab689b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-13 20:51:20.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/8/check/dsvm-tempest-cisco-zonemanager-job/fc9b6ea/ : FAILURE in 33m 31s","accounts_in_message":[],"_revision_number":8},{"id":"c68e1dda8e9dca0201d778453e7a9151554ee735","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-13 21:33:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/8/upstream-solidfire-check/cinder-solidfire-iSCSI/991c7cf/ : SUCCESS in 2h 02m 24s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"825dba8685d9eccdb272a1db4b26dcc8b20f1afd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-13 22:02:14.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/8/upstream-check/cinder-cDOT-iSCSI/aaf318a/ : SUCCESS in 39m 01s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/8/upstream-check/cinder-cDOT-iSCSI-IPv6/ea9998f/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/8/upstream-check/cinder-cDOT-NFS/1fd2d8d/ : SUCCESS in 36m 35s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/8/upstream-check/cinder-cDOT-NFS-IPv6/624cbb4/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/8/upstream-check/cinder-cDOT-FCP/ba0c4cf/ : FAILURE in 2h 31m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"8e6b190706ec514949867389395c75288278f407","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 22:24:06.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"561e1adf806531ddc02e3f08781b335cef2d55ff","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-13 23:51:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/8/check/dsvm-tempest-zadara-driver/5cfe53d : SUCCESS in 1h 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"388d7e15f9f04b8f6fbadfbbf29fc4091e1d554f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-08-14 00:18:06.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/07/737707/8/check/DellEMC_SC_FC/5001fae/DellEMC_SC_FC/295 : SUCCESS in 1h 27m 11s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/07/737707/8/check/DellEMC_SC_iSCSI/f048ce4/DellEMC_SC_iSCSI/269 : FAILURE in 1h 31m 17s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"08242d1fcbdc8bf861a927e708a440eb165c3c6a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-14 00:18:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/8/check/DellEMC_XtremIO_FC/e80cdd0/DellEMC_XtremIO_FC/None : EXCEPTION\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/8/check/DellEMC_XtremIO_iSCSI/a2d9c1f/DellEMC_XtremIO_iSCSI/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"cb6b78692c614c9a0074a1872eb205ac8db6c360","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-08-14 00:54:53.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/8/check/DellEMC_PowerMAX_FC/a1fc033/DellEMC_PowerMAX_FC/456 : SUCCESS in 3h 46m 03s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/8/check/DellEMC_PowerMAX_iSCSI/4f4496f/DellEMC_PowerMAX_iSCSI/388 : SUCCESS in 4h 38m 09s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"50b3b30889b5bb88d4f13a2524f9683d64b8c781","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-08-14 01:32:35.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-8/Seagate-iSCSI : SUCCESS in 49:20.67\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-8/Seagate-FC : SUCCESS in 49:03.54","accounts_in_message":[],"_revision_number":8},{"id":"c629d37830f0140ee0eb2949af0387f28c783cf5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-08-14 01:42:29.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/07/737707/8/check/DellEMC_PowerFlex/a5562d4/DellEMC_PowerFlex/370 : SUCCESS in 2h 08m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"5d4e4a7fb7ad8ec162d405b276dd2c445741ac23","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-08-14 01:52:52.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-8/Lenovo-iSCSI : SUCCESS in 48:46.03\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-8/Lenovo-FC : FAILURE in 49:22.99","accounts_in_message":[],"_revision_number":8},{"id":"a1b593ddcc54bbd92bb6102977f3fdd95b63312f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-08-14 01:57:40.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/737707-8/HPMSA-iSCSI : FAILURE in 50:48.97\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/737707-8/HPMSA-FC : SUCCESS in 45:35.14","accounts_in_message":[],"_revision_number":8},{"id":"26c1a3bbe9f0fc2f655698cafac4f97b58343d48","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-14 03:38:03.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"2b92351ff4753fd9350956536e0ee2ab66e7465a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-14 04:36:07.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/07/737707/8/check/DellEMC_VNX_iSCSI/d26bd29/DellEMC_VNX_iSCSI/264 : SUCCESS in 2h 49m 27s\n- DellEMC_VNX_FC http://publiclogs.emc.com/07/737707/8/check/DellEMC_VNX_FC/8259923/DellEMC_VNX_FC/261 : SUCCESS in 1h 03m 03s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"9f0eca6e5b0820c17a65d550d132b8bd6d3757da","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-14 07:19:00.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"67137b7b39478350d719bb76a7897c9701115b30","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-14 15:57:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1460/index.html : SUCCESS in 1h 17m 43s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1607/index.html : SUCCESS in 1h 37m 14s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1629/index.html : SUCCESS in 1h 23m 49s","accounts_in_message":[],"_revision_number":8},{"id":"5de73f188ebc610802758f499b5f02bcf85bb07d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-17 05:32:25.000000000","message":"Patch Set 8:\n\nrun-HPE Storage CI","accounts_in_message":[],"_revision_number":8},{"id":"600b81e0de9c747def616f5323e5710b6ddda45f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-17 06:46:41.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e016277/e016277.html : FAILURE in 1h 11m 01s","accounts_in_message":[],"_revision_number":8},{"id":"bb62be750b2af1749ab56829d4ea0d04c5da0ac5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-08-17 15:34:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/8/20200817214723/NEC-ISCSI : SUCCESS in 2h 25m 51s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/8/20200817214723/NEC-FC : SUCCESS in 2h 47m 22s","accounts_in_message":[],"_revision_number":8},{"id":"e3be9419a681568a7d35875e58191bcb45052d1a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-24 06:01:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ea852d287889dfed876e1c4314bd11e9a42ccbf3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 06:04:11.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"fd2c5def73f69b3cf51bfcfe8b8a75f7e3532422","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 06:06:06.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/9/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/d6817d9 : FAILURE in 4m 19s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/9/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/c0fcee7 : FAILURE in 4m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"c0d9c234c8e20aa5e53c564a49b001b553926e1e","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 06:13:17.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/8a564f1b3e125276e8c2768e878e88ed23215b11/index.html : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"d604c4def9f2e497619f409971190205b671221e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-24 06:40:05.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/9/check/dsvm-tempest-cisco-zonemanager-job/0a9fd50/ : FAILURE in 33m 00s","accounts_in_message":[],"_revision_number":9},{"id":"76c3ade9eea5e06176f3174a5d8ce477c0663cf9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-24 06:40:24.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/9/check-not-vote/ext-cinder-zuul/0314a5b : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":9},{"id":"de1266acc7d998c9cbcccd05d96ac550c9fee155","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-24 06:43:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/9/check/dsvm-tempest-zadara-driver/308b6ac : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"760f5e158a7fcc665759c69096f3219660a8bc17","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-24 06:52:25.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"20552a9593bcd86c5642ec8796f34fef43e98006","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-24 06:54:26.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/9/iscsi/ : FAILURE in 52m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/9/smb/ : FAILURE in 52m 33s","accounts_in_message":[],"_revision_number":9},{"id":"872ad58a829d59fe2662f160644f6acd21c8812d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-24 06:57:07.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/140ec89/140ec89.html : FAILURE in 51m 45s","accounts_in_message":[],"_revision_number":9},{"id":"5fd9257dfe12996fe4ff80da0f4453b056a9e7b1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-24 07:03:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"080fb2f279ad19a3b7b3d21d9be928b2b0ed58b1","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 07:08:40.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/10/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/bdc8d66 : FAILURE in 4m 23s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/10/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/cc483a2 : FAILURE in 4m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"607662ed7759feb43cfb456637f831ee06ae06f4","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 07:15:49.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/149f08f352602678af3a4f9b4144a9b86c2ccbf3/index.html : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"efa5f5f049006a17a3b3569e50f1d419551a333d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 07:30:06.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"2e2b550d8af2b95bbe6076a3ebefcc9c29cdfe34","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-24 07:41:51.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/10/check/dsvm-tempest-cisco-zonemanager-job/3a1acec/ : FAILURE in 32m 51s","accounts_in_message":[],"_revision_number":10},{"id":"8d28c4011d14144045e6b54c0d7d6a465d7cfd3a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-24 07:45:27.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/10/check-not-vote/ext-cinder-zuul/5650749 : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":10},{"id":"8ed8eb6a243e8c0952ddbe605e98475e85f62721","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-24 07:54:39.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"22161d2e0157bf128462f6c3b720c8e008383af5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-24 07:56:53.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/10/iscsi/ : FAILURE in 52m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/10/smb/ : FAILURE in 46m 54s","accounts_in_message":[],"_revision_number":10},{"id":"0c300a1b0295931b6832d4dc8bf791aa1a0a3830","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-24 07:58:31.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/8701844/8701844.html : FAILURE in 51m 38s","accounts_in_message":[],"_revision_number":10},{"id":"c78d2003b686b3f6133540ef718bf5c436fb0de7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 08:05:50.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"453c90986bdbd8f504438f5cf12ff7f2ab095a2c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-24 08:07:51.000000000","message":"Patch Set 9:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-9/ns5_nfs : SUCCESS in 2h 4m 44s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-9/ns5_iscsi : SUCCESS in 2h 6m 3s","accounts_in_message":[],"_revision_number":9},{"id":"5eef2e169ad3936d2c0e2fe8c888ae9a6bac1dca","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 08:30:53.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/9/20200824150127/NEC-ISCSI : SUCCESS in 2h 10m 34s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/9/20200824150127/NEC-FC : SUCCESS in 2h 29m 24s","accounts_in_message":[],"_revision_number":9},{"id":"9dd574886333efe9a553174b787711bc1ea35b56","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 08:40:56.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/10/check/DellEMC_XtremIO_FC/0da03d2/DellEMC_XtremIO_FC/759 : SUCCESS in 1h 32m 54s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/10/check/DellEMC_XtremIO_iSCSI/db5885f/DellEMC_XtremIO_iSCSI/726 : SUCCESS in 1h 36m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"5fb73ed7ed4b84219a8a67b9adaf5a476c0d8d89","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-24 08:41:19.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/10/check/tempest-dsvm-full-bionic/663bae0/ : SUCCESS in 1h 19m 22s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/10/check/devstack-plugin-ceph-dsvm-bionic/30285f2/ : SUCCESS in 1h 28m 13s","accounts_in_message":[],"_revision_number":10},{"id":"f798b59891fa81ea981ba31bfa15d348f8a32c0d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 08:43:18.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"8b237375d3de6b2b1b132d48ddf375746930f8fa","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-24 09:08:12.000000000","message":"Patch Set 10:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-10/ns5_nfs : SUCCESS in 1h 51m 25s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-10/ns5_iscsi : SUCCESS in 2h 3m 54s","accounts_in_message":[],"_revision_number":10},{"id":"c58ded6a617958425dac3aa7d6f26c8b97e0095b","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-24 10:10:47.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/10/check/nimble-iscsi-driver-dsvm-volume/ecf7033 : FAILURE in 3h 06m 37s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/10/check/nimble-fc-driver-dsvm-volume/0aefbcb : SUCCESS in 2h 46m 23s","accounts_in_message":[],"_revision_number":10},{"id":"94c2baf1f1c8c12948c636e7b7536ce12d19a49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 10:15:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/929434fdf3b3474ab1b982841cce43b1 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b71f691478bd43a48af84cb8f93c8645 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5df63a63124a6cabf60c6feb30c4fb : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9d360b2876a49c49f64f04873badb5f : SUCCESS in 12m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e39e06e3d3b4e35b423aca5d78d1491 : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4002a044424146b6847ab6ca67f7c5 : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/7914af9f22614c10829edfed1ef51258 : SUCCESS in 1h 04m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6b3004a5bb947ccbcca6ed8667b12e1 : SUCCESS in 1h 15m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/449d3592e5334a9f92b68994c16654bf : SUCCESS in 7m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b98a082fe4684c049023fce4f80a0077 : SUCCESS in 13m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5348d4eeab274642b34d33396323a214 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b312f1fe7cf546fc844eb50c5e128e3f : SUCCESS in 6m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a577c79c1f5146d39d25a35feb9739bc : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bd3a50b37dae4fe9b0e6d99c523772dc : SUCCESS in 39m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ddbde6220cb428f9e9bf70b6861f883 : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d84b88b16b8347a69281facd91c66021 : SUCCESS in 1h 18m 16s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d6caf4060df4657bfe5e20f018fa5f4 : FAILURE in 1h 27m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/747ef35ab65b429b98d237af41caa5a5 : SUCCESS in 1h 09m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e10d03071734c53adec52263cf99276 : SUCCESS in 1h 22m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4abfb44a1a92421786dc6ac7203325c0 : SUCCESS in 1h 22m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc14108cd35745a0a26a96769acf2e29 : SUCCESS in 2h 00m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/022e268c32b149439d8af91a14d1dcc4 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":10},{"id":"670ff5748c9eabe6e5c5cfe8a4defa98f45f5d46","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 11:01:14.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/10/20200824173057/NEC-ISCSI : SUCCESS in 2h 18m 42s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/10/20200824173057/NEC-FC : FAILURE in 2h 30m 15s","accounts_in_message":[],"_revision_number":10},{"id":"385e29108881752f3bb76f74cf1c94a7284ef0db","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-24 11:33:21.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"12edcbcd179cec90159de0b471f6b4ab4ec38d9b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 11:38:06.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/10/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9194da2 : FAILURE in 4m 21s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/10/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/1cea1ab : FAILURE in 4m 28s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":10},{"id":"37be72e42235b43fb7d4d4cf8805f721be78d450","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 12:00:27.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"7277e7a53a63d0a6f7b822a491f6d9400ab0bf76","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-24 12:11:34.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/10/check/dsvm-tempest-cisco-zonemanager-job/edd9f0a/ : FAILURE in 33m 12s","accounts_in_message":[],"_revision_number":10},{"id":"ee5f82470397d9db2727a130c28916bf805cc61b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-24 12:12:21.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/10/check-not-vote/ext-cinder-zuul/dbeac14 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":10},{"id":"b1497476fd5d5db45cd727682f7a369b8bc9e98c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-24 12:13:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"022e68152f1f5416ccf5a3298a63fa46b4c8bf46","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-24 12:13:45.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"6408c6aed28699421a6f22e11f9ca13984d15b53","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 12:18:12.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/11/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/64838f7 : FAILURE in 4m 35s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/11/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/e64e44f : FAILURE in 4m 23s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":11},{"id":"61896ab6963a7db23092a325f5176e22bf9042f2","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-24 12:24:18.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"e2b50042222d5fb9e6a446fa3fb3aadba2516b15","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 12:24:46.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/47f911b742d87bf61ba7720cb90cd47582c47de0/index.html : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"7d1948d1341b3e4230bd8ccf5c6cdfc09ee6f0c9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-24 12:51:32.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/11/check/dsvm-tempest-cisco-zonemanager-job/a65722f/ : FAILURE in 33m 15s","accounts_in_message":[],"_revision_number":11},{"id":"c7405ec64298a3c56b1a9469e7f8bb9a2e69d257","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-24 12:52:57.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/11/check-not-vote/ext-cinder-zuul/657d557 : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":11},{"id":"41c1efe90a8482e6d1a67d08b816e0bd5ddc07a0","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-24 13:04:16.000000000","message":"Patch Set 11:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"c31846c31291652fca05e12e7dad195bd9ab9530","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-24 13:05:35.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/11/iscsi/ : FAILURE in 51m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/11/smb/ : FAILURE in 46m 07s","accounts_in_message":[],"_revision_number":11},{"id":"3e195fc76484b23a8e205dfe9a35159f5431bb9e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-24 13:08:01.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1f4eac1/1f4eac1.html : FAILURE in 52m 29s","accounts_in_message":[],"_revision_number":11},{"id":"c2de4208fcb32f0726e5674d1f46372e10ad14e1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 13:13:50.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"60ff0fb1d190b6c447167cfe386fcd7cd0f20f67","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 13:50:51.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"dc6befb12ab6b1a3d41b3adcd800b17db11dfdf1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-24 14:07:34.000000000","message":"Patch Set 11:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-11/ns5_nfs : SUCCESS in 1h 48m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-11/ns5_iscsi : SUCCESS in 1h 53m 53s","accounts_in_message":[],"_revision_number":11},{"id":"0c27d234642596da070ecc3c9194f4f083797434","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-24 14:14:32.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/11/check/tempest-dsvm-full-bionic/ad3d419/ : SUCCESS in 1h 25m 53s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/11/check/devstack-plugin-ceph-dsvm-bionic/74d6e64/ : SUCCESS in 1h 52m 46s","accounts_in_message":[],"_revision_number":11},{"id":"cb848382c369ce5f3441506d6d44ca32d8388c37","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 14:16:50.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/11/check/DellEMC_XtremIO_FC/a616ebb/DellEMC_XtremIO_FC/762 : SUCCESS in 1h 35m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/11/check/DellEMC_XtremIO_iSCSI/2d23501/DellEMC_XtremIO_iSCSI/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":11},{"id":"2715d3014cc07306cb48641ac5fe0d5ea0ac2a61","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 15:05:09.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"b3f2458a9a255487c0d93585c49479c8cc8d415f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 16:33:51.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/076494e4c26f48b7a0421c1c67232eaa : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6e77a0306f9415ca2a62837ef46c7a5 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ef2e42b40f43ca85008b03657d0643 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/581b1e013d084ad59d1d71e3794074ba : SUCCESS in 12m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86683c0604984192bbf989d75d2b7e2e : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e596bcd2919b4abe9786588243d6737f : SUCCESS in 14m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/f7533761b31a42e2a2c9d3372cf8cea7 : SUCCESS in 58m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/58360e63ce45440c9edfe346cc98d4b1 : SUCCESS in 1h 17m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f427ff529cd43a08ae0336256464b8c : SUCCESS in 10m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/582b8f0f10b54265a0623f45cc1036b1 : SUCCESS in 11m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2d108b71553456aacb803720c06332a : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f57fddb720e145faa6a5947b6525b356 : SUCCESS in 5m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bdd000397a624c688ef5ff8939bad164 : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3c6d0dffcb424ebba6d48cc45669277a : SUCCESS in 42m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0b43b82ae45d473e8ca55456ae7da4c5 : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/76fce1309a8a4f349004ced6ef63f83f : SUCCESS in 1h 07m 52s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f48cfe2ec0cb4eacb6cedfd4510a52c3 : SUCCESS in 1h 31m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a8bea792240140c1830a993e7275fd16 : SUCCESS in 1h 06m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8ca76f5dd425488eb8d387f4bb06f084 : FAILURE in 1h 13m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/87ad40d2f17840ec85846ee3378209bc : SUCCESS in 1h 27m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02b107361fd6448193ae1de425ec1fdf : SUCCESS in 2h 17m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df8548496357482882d3623b5d975216 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":11},{"id":"90e89ebf7041b711707f3600cbdd1084b70954f3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-24 16:58:38.000000000","message":"Patch Set 11:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-11 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":11},{"id":"fea141b885a205505f5dcb103b4251fff891c768","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-08-24 17:02:56.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/11/check/synology-dsm-cinder-iscsi/8e90022 : FAILURE in 28m 02s","accounts_in_message":[],"_revision_number":11},{"id":"f2d6fc5a1aa5d2ce4d004b7f5ef177105a3a1660","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-24 17:44:24.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/11/check/DellEMC_Unity_FC/1ce8383/DellEMC_Unity_FC/489 : FAILURE in 2h 49m 18s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/11/check/DellEMC_Unity_iSCSI/77572ca/DellEMC_Unity_iSCSI/461 : SUCCESS in 1h 38m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"eba0f0a558a0072da5c716d55d63a3a42d1ffddb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 17:44:57.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"6b16ab5384a6d15c1bdfa2c7349ee1e0628fe2a3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 17:46:32.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/11/upstream-check/cinder-cDOT-iSCSI/8e8fb91/ : FAILURE in 1m 45s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/11/upstream-check/cinder-cDOT-iSCSI-IPv6/b5af37f/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/11/upstream-check/cinder-cDOT-NFS/1025d89/ : SUCCESS in 37m 21s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/11/upstream-check/cinder-cDOT-NFS-IPv6/a72b062/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/11/upstream-check/cinder-cDOT-FCP/bc084ca/ : SUCCESS in 1h 07m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"6aa035483989718c21591e2f32ce33c296a784b5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 18:19:27.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/11/20200825004350/NEC-ISCSI : SUCCESS in 2h 18m 52s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/11/20200825004350/NEC-FC : FAILURE in 2h 35m 34s","accounts_in_message":[],"_revision_number":11},{"id":"b1e252cbc6169db4886f1e53c1b1065e4b75cddb","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-24 18:20:10.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/11/upstream-solidfire-check/cinder-solidfire-iSCSI/f5aa77a/ : SUCCESS in 1h 41m 09s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":11},{"id":"ec45730992647564256a8d25ede0e14714383226","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-24 18:36:39.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/11/check/nimble-iscsi-driver-dsvm-volume/5bf54e4 : SUCCESS in 2h 45m 07s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/11/check/nimble-fc-driver-dsvm-volume/943acfc : SUCCESS in 2h 44m 15s","accounts_in_message":[],"_revision_number":11},{"id":"9bdb9076e6ef11372b630dbacba676d31a3e7042","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-08-24 18:40:59.000000000","message":"Patch Set 11:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-11/Seagate-iSCSI : SUCCESS in 44:41.94\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-11/Seagate-FC : SUCCESS in 42:58.19","accounts_in_message":[],"_revision_number":11},{"id":"08190c2084e3d7b12d714b43a6e34b16610d478c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-08-24 18:44:59.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/07/737707/11/check/DellEMC_PowerStore_iSCSI/8d4886c/DellEMC_PowerStore_iSCSI/147 : FAILURE in 1h 02m 49s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/07/737707/11/check/DellEMC_PowerStore_FC/86b6399/DellEMC_PowerStore_FC/148 : FAILURE in 1h 04m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"d166b98a920787a78972116600c8e2eda2d0becd","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-08-24 18:50:16.000000000","message":"Patch Set 11:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-11/Lenovo-iSCSI : SUCCESS in 49:58.77\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-11/Lenovo-FC : SUCCESS in 47:33.04","accounts_in_message":[],"_revision_number":11},{"id":"f3ba7aa204d72e72ac1b0bebb63e1963ca0ea4e8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 18:52:10.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"17f7384db2b86a4a9c88cac43e4cccbdf031e32f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 19:54:00.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/07/737707/11/check/DellEMC_PowerFlex/566a717/DellEMC_PowerFlex/429 : SUCCESS in 2h 08m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"ad6cfeeb7a3b5883c0522a78eb1f9cd8f549bee2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-24 20:28:04.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"295d808d451312dfdfcdde3e7b2dc37312cf2da2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-08-25 02:11:48.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/11/check/DellEMC_PowerMAX_FC/ac5fe1f/DellEMC_PowerMAX_FC/524 : SUCCESS in 4h 12m 26s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/11/check/DellEMC_PowerMAX_iSCSI/09947ac/DellEMC_PowerMAX_iSCSI/451 : ABORTED in 3h 22m 00s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":11},{"id":"4f5b7bba4cb0450ede261bce8c97a791728e63c0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-25 02:44:56.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"a9215f617211f9220cf71ab6d1692a6a59076c3a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-25 03:43:55.000000000","message":"Patch Set 11: Code-Review-1\n\n(21 comments)\n\nI haven\u0027t reviewed the tests yet, but have some comments for you to look at inline.","accounts_in_message":[],"_revision_number":11},{"id":"7794b35cf0b88fa4b5940f43d4a9ca3ffccee45e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-25 12:57:51.000000000","message":"Patch Set 11:\n\n(14 comments)\n\nLeft some comments about the api-ref.","accounts_in_message":[],"_revision_number":11},{"id":"93fa169d8609df7e94f6323e182d666468337fa5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-28 14:30:22.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"33e26bf047a89e6d8203ee916780fd67c0c2f367","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-28 14:42:46.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/3d0e37fd963be00fb5a3348f1de16c88161ae3e8/index.html : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"eed47650a5e851b3d4a5c3a252a7a35bfafab6c2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-28 15:17:52.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/12/check/dsvm-tempest-cisco-zonemanager-job/b12302d/ : FAILURE in 33m 13s","accounts_in_message":[],"_revision_number":12},{"id":"077314f5fb0f14df2e4b3fc314bdc82f421c8d5f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-28 15:21:37.000000000","message":"Patch Set 12:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"72bfd5eb788019905dbc479b2f017e69fb6b311d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-28 15:26:36.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/12/iscsi/ : FAILURE in 55m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/12/smb/ : FAILURE in 48m 14s","accounts_in_message":[],"_revision_number":12},{"id":"2e2263eaf0b9e506e99b38b6a1116e641dd8484c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-28 15:27:09.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/52f50aa/52f50aa.html : FAILURE in 53m 21s","accounts_in_message":[],"_revision_number":12},{"id":"fc6a651c81bf3e5781e975675210b4cc7f56ba68","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-28 15:27:30.000000000","message":"Patch Set 12:\n\n(34 comments)\n\nThanks for the detailed review Brian.\nI\u0027ve some replies inline.","accounts_in_message":[],"_revision_number":12},{"id":"2d4e284b042ca68016577d1b152838802a5e7c66","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-28 15:46:05.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"f4074adbb42dbf8c2ed2b3567dcb18af54e064f3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-28 16:05:10.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/12/check/DellEMC_XtremIO_FC/cdb5b6c/DellEMC_XtremIO_FC/768 : FAILURE in 1h 34m 31s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/12/check/DellEMC_XtremIO_iSCSI/37aca39/DellEMC_XtremIO_iSCSI/734 : FAILURE in 1h 34m 09s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"a16bad12e1f69aee000ecd2bc36c7fa02803b7f4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-28 16:14:12.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"efc7ae5af8b3a33132af3493bbfd1b9dd823d3d2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-28 16:20:27.000000000","message":"Patch Set 12:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-12/ns5_nfs : FAILURE in 1h 47m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-12/ns5_iscsi : FAILURE in 1h 49m 38s","accounts_in_message":[],"_revision_number":12},{"id":"9a3ab3b12a5797334ea3f606b79d9ec06037ccbd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-28 16:26:12.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/12/check/DellEMC_Unity_FC/4a847f9/DellEMC_Unity_FC/518 : FAILURE in 1h 32m 43s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/12/check/DellEMC_Unity_iSCSI/2c9808c/DellEMC_Unity_iSCSI/499 : FAILURE in 1h 50m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"309edf59b860d7fcf6673aeb9ab55fad04870760","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-28 17:17:41.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"bcaa78b3acc2929ce71f6d6b74e488f4484694a2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-28 18:29:03.000000000","message":"Patch Set 12:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-12/ns5_nfs : FAILURE in 2h 3m 32s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-12/ns5_iscsi : FAILURE in 2h 5m 11s","accounts_in_message":[],"_revision_number":12},{"id":"b76b8617eda1b354331f465c555d6eab456b0271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:43:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/461d67785f2f4ef5a8f264b120f8fabe : SUCCESS in 7m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a7120e8d2ba4c7bb9f3b4a9fdae2633 : FAILURE in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f801a935d4914c84ba2325014b66baef : SUCCESS in 11m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4deaee0ba5374a2aafc524647737b2a1 : FAILURE in 13m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da4c266ffcde4773b330ac84c1ba9dc5 : FAILURE in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b1387f7f5fc42658f532e1ed44b8a5a : SUCCESS in 16m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/de7c685a3a6b42a3b054c20ee8dcbd74 : FAILURE in 54m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cfeb166afef4cc0823a91e929afb844 : FAILURE in 57m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63a8610fa69e4621ada5259c63b3bdca : SUCCESS in 8m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f1462a206937493aa5f1a211530d06f2 : FAILURE in 10m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6238263a68f745eabe6101203c760882 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a9f128aec5c44550a3143ca0bec0643d : FAILURE in 7m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3065bcdbe5524cb38ce9426bc5eef6aa : FAILURE in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f26c763d5e6341dda57d9d76f53ee7cb : FAILURE in 47m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/537f9ba46fad49f19ec307f8ac990b6f : SUCCESS in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/06979ae00114424a8b7acb01434df93b : SUCCESS in 1h 13m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5b07f91844044d2686d5b1a83175d606 : FAILURE in 1h 13m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a3fcc9adb5cc49aa85fab575cecde0fd : FAILURE in 1h 25m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e9fe701b40644e48a80940ed9405d6f4 : FAILURE in 1h 00m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/adc25c0e5775423d871440621fe2dfa0 : FAILURE in 1h 32m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2464bd707ec94ec2b90e803791ab777b : FAILURE in 1h 13m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dcd754238d194f04b1863f1abbc74bca : FAILURE in 2h 01m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6d45dc8113c43309482f97931c19518 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":12},{"id":"3a43f1dd96f8614884dbc8808c7f52c3905201a2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-28 20:24:20.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"94d0f2ed729d3c7df1321f7b70095ec266de0175","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-28 20:57:26.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"fccc82523866b8a6909827df2ae0bc8217604979","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-28 21:02:06.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/13/check/dsvm-tempest-cisco-zonemanager-job/442767f/ : FAILURE in 32m 28s","accounts_in_message":[],"_revision_number":13},{"id":"e4fd1295a92d8c586ed833edc963a1a79734c3e5","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-28 21:15:41.000000000","message":"Patch Set 13:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"2b35702f6bd9bffd582b00d33bbf65b8c39028c2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-28 21:23:57.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/b9103ed/b9103ed.html : FAILURE in 54m 04s","accounts_in_message":[],"_revision_number":13},{"id":"671acad2bd012627cf9e8a1baa3bb6442b1cd1e8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-28 21:26:33.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/13/iscsi/ : FAILURE in 46m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/13/smb/ : FAILURE in 1h 01m 38s","accounts_in_message":[],"_revision_number":13},{"id":"1c84e3ddba0db23b84d1d52337a3188c36b1fa48","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-28 21:28:38.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/cf34f6cb39b5165bdcd8f88af5e9586d7bb6a7ff/index.html : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"e24928b85ef10178aea8783136833573d7da3e3e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-28 22:11:09.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"93aafac85bdbac7ba084751f3744c66feb0fffe7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-28 22:27:55.000000000","message":"Patch Set 13:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-13/ns5_nfs : SUCCESS in 1h 49m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-13/ns5_iscsi : SUCCESS in 2h 3m 11s","accounts_in_message":[],"_revision_number":13},{"id":"6d5c02e19c2bf11d08b90f6fe5df74e77020f11d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 23:59:55.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/718e72a1dfbe429f9978ac73ea976560 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a20f764ceb64178894031c10456644b : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c7fb1e27a5d4575be34aafcf30a4eaa : RETRY_LIMIT in 0s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72d2ad9a6b11441399acdd9f2b07d5bd : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1e69222ebf04637827b8862c6e81988 : RETRY_LIMIT in 0s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5992843f2a734904960947a078b3bcda : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/edb87cfd62e14f669e126552140b8d3c : SUCCESS in 59m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b9a4dc1d762e4688b0c1887b2e64df12 : RETRY_LIMIT in 0s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9efa4994feef43c695c0c80092fca927 : RETRY_LIMIT in 0s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/37f27463e5a54783bee2d20304d71c91 : SUCCESS in 11m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dcf286014874a2fbea0cd60e00a0c00 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3337f4f566848488a0e7817aece3387 : FAILURE in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/832d0414e8a44ef0a5ae8db7e57e6e9a : FAILURE in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9c7a244ed3cd4616b456c8051b337f92 : SUCCESS in 1h 06m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/228d875d7d8b44d4b3dc524c86927f03 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b19ee5d80624052b45f43a8299093f3 : RETRY_LIMIT in 0s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e54d1c74610496496fc34680876324e : FAILURE in 1h 04m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5376a264cb454c619f36c1d93c5df5ce : RETRY_LIMIT in 0s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a52bceaf24dc4ae790c8abc585227a0c : SUCCESS in 1h 03m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/798ca378103948c5b578559e19b6392a : SUCCESS in 1h 14m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/712440147ba34eb4ab8f91b376dfbaf3 : SUCCESS in 1h 25m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/80658cce6a134ecab61ce9d210ca2da4 : FAILURE in 2h 25m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cad226af2f954bc6a6bebfc127e07c77 : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":13},{"id":"5e0af7e6c6642570cd26ab0d01d35218d895139b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-29 00:15:26.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/13/check/DellEMC_XtremIO_FC/9c1bbd3/DellEMC_XtremIO_FC/773 : SUCCESS in 1h 35m 34s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/13/check/DellEMC_XtremIO_iSCSI/f92f6bf/DellEMC_XtremIO_iSCSI/739 : SUCCESS in 1h 35m 21s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":13},{"id":"48178d80bd556e2b7ab17bb6ea5c639ec2516c29","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-29 00:58:50.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"c440fee291042f0f3d12c74d439805860fd714b6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-29 01:07:11.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/13/check/DellEMC_Unity_FC/1b424d7/DellEMC_Unity_FC/525 : SUCCESS in 1h 19m 10s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/13/check/DellEMC_Unity_iSCSI/36a9a17/DellEMC_Unity_iSCSI/504 : SUCCESS in 1h 46m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"df7c5cb124a43786171e4c065b7f710601dd5892","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-29 01:52:17.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"f10e58def454facda8dc8e46e31d32afe4028897","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-29 02:15:07.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"035e4b8c53e910e4879acfef3bb6d481796b2576","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-29 02:53:28.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/737707/13/check/dsvm-tempest-cisco-zonemanager-job/4fc2806/ : FAILURE in 33m 06s","accounts_in_message":[],"_revision_number":13},{"id":"6330ad1b20472e4bf4c0bc4132ba9b3b0c277f35","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-29 03:04:21.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"329da4e506b828245b899fbfe928a35d53bf8ee6","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-29 03:04:42.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/13/iscsi/ : FAILURE in 47m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/13/smb/ : FAILURE in 49m 09s","accounts_in_message":[],"_revision_number":13},{"id":"3c9be5d927d05d99a233dd33d3af15d1e8059b51","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-29 03:06:28.000000000","message":"Patch Set 13:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"9c3ee6042859c0b9bcc80e4662937083a0427687","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-29 03:39:06.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"2b55b596e2274f49bcdf833c841c4aa4bb9c8097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 04:49:42.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3cf16f698d6946d681d1bf8eb2da7ef5 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7daa7cb6f6a34617b15f6b7bd5ebce6b : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd9b6e249254d94926acb6b1fa6feb1 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57e4cbfe49b944c59d3b15d602c3f380 : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89201198888b47e5baebb81bffc35742 : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a103a9c0e20b442e91f99e1977af8dbd : SUCCESS in 27m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/b9350374ecb84f4ca04c93fc1490f252 : SUCCESS in 55m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c26be3a6b2e4508b2a33bb0460f14cc : FAILURE in 2h 05m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9cbb5cec988460c967d47d61c595470 : SUCCESS in 12m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/62a92f8323714ef990bf96eabcffab78 : SUCCESS in 12m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf5addd749374f1596eb1d38354a61fd : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/333e95a4790e404e955fb6f2446f8574 : FAILURE in 14m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76bbead62c974ea8ba087dc51f064057 : FAILURE in 15m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/613b2881fa77426faa805947a0e213cd : SUCCESS in 1h 14m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aeb8a677b67f47568584f8443c57b757 : SUCCESS in 13m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2ff2c52626f844c88252304a0f588b29 : SUCCESS in 2h 03m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d2d8f3ddca64067a8d73a156f93990c : TIMED_OUT in 2h 05m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d6220ce755d4e779e4a155fb7aa52e0 : SUCCESS in 2h 31m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/31af221d11a24d569e565d8b0cf43b4d : SUCCESS in 2h 09m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/93ee6d832f6c4458a700389a44aff0b1 : SUCCESS in 2h 14m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/94c22efab7734d3ba3b013eccf994e82 : SUCCESS in 2h 22m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/188a2f79244d45dba92c4b12f81b3b26 : SUCCESS in 2h 03m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a66267e1d57435f8eb5246611973727 : SUCCESS in 1h 51m 41s","accounts_in_message":[],"_revision_number":13},{"id":"da0a7016631dff73c7706cea357abde9edb03dd4","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-29 06:33:40.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/13/check/nimble-iscsi-driver-dsvm-volume/1e7b551 : FAILURE in 17s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/13/check/nimble-fc-driver-dsvm-volume/dd098dc : FAILURE in 8s","accounts_in_message":[],"_revision_number":13},{"id":"59e32b1c234de89755923441cf6bb2625eb9fafa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-29 08:38:52.000000000","message":"Patch Set 13:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/13/check/tempest-dsvm-full-bionic/9afc48e/ : SUCCESS in 1h 37m 04s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/13/check/devstack-plugin-ceph-dsvm-bionic/7b79abd/ : SUCCESS in 1h 39m 31s","accounts_in_message":[],"_revision_number":13},{"id":"3aa978b4952d1c6ca9e9dc89895261d2c0b63d52","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-29 09:14:16.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"df180134a25de77964a60e28547e989f6c611c74","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-29 12:53:57.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- ext-cinder-zuul : EXCEPTION","accounts_in_message":[],"_revision_number":13},{"id":"3d005317d62884a64af10a9a7dc36f5ad462c3ed","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-30 01:55:57.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1621/index.html : SUCCESS in 1h 21m 32s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1765/index.html : FAILURE in 18s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1779/index.html : SUCCESS in 1h 20m 44s","accounts_in_message":[],"_revision_number":13},{"id":"e549ebf892165230006558fc89ae6cf153600422","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-30 02:29:24.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"f57c4fb2ac22fac3ad27e03f8ddee7ffd62fba07","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-08-30 18:26:44.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/13/check/DellEMC_PowerMAX_FC/16a1a31/DellEMC_PowerMAX_FC/558 : SUCCESS in 3h 54m 08s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/13/check/DellEMC_PowerMAX_iSCSI/7526d8e/DellEMC_PowerMAX_iSCSI/485 : SUCCESS in 4h 53m 40s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":13},{"id":"56ff7b3a9c5bd8690347ae2c44b240a20b85b194","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-30 19:01:13.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"34667789db70725cba173f1cd1f09baee5ee74af","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-31 04:30:19.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1637/index.html : SUCCESS in 1h 27m 27s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1781/index.html : SUCCESS in 1h 41m 09s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1795/index.html : FAILURE in 25m 31s","accounts_in_message":[],"_revision_number":13},{"id":"948ac47f0b0055e20211b6c24234ac2940772364","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-31 05:05:02.000000000","message":"Patch Set 13:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"fb26e45051ae0bde5bf1fe172eb86d223eb60631","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-31 09:53:31.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"265e62f50828bbc5c7cd5faa38a4489d00df73e6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-31 09:54:26.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"696bbe0c10b6823d49d231ee0a07946d86f45b07","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-31 09:54:54.000000000","message":"Patch Set 14:\n\nBuild Unstable \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/82214b5d3317b0913139e184568f15c4649db971/index.html : UNSTABLE","accounts_in_message":[],"_revision_number":14},{"id":"5c87ce5683167580d668247c312653530c06551c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-31 10:06:37.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/14/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c0b1353/c0b1353.html : FAILURE in 9m 48s","accounts_in_message":[],"_revision_number":14},{"id":"38bde2bbbc3da5825a9d7fcb82d0853588289d16","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-31 10:10:14.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/14/check-not-vote/ext-cinder-zuul/1d47c8b : FAILURE in 15m 13s","accounts_in_message":[],"_revision_number":14},{"id":"934446f254a9e2fc60e4c6fa83da160791fa1c56","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-31 10:14:54.000000000","message":"Patch Set 14:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-14/ns5_nfs : FAILURE in 20m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-14/ns5_iscsi : FAILURE in 18m 29s","accounts_in_message":[],"_revision_number":14},{"id":"b5453d29062633f1dc95983cb5fb6e66989b5d21","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-31 10:22:22.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/14/iscsi/ : FAILURE in 22m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/14/smb/ : FAILURE in 27m 28s","accounts_in_message":[],"_revision_number":14},{"id":"143636476141eb939126e298b1924caa15ae66f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 10:29:17.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2a56b96a36554a9b9713d340f49ce94f : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d450c54f7ee34dd1b34a408d756ace5e : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a554745c8b43398bd92809b5adb797 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09a8d27611ed445e9928b65684cf7600 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf704e9d5d7041cbb12c5f6db3d3a841 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dd875972d144b28010ee071e8f29d3 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/a7803ea207024470aff44e8e04cd3b97 : FAILURE in 18m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c7eb51b51b74aa4a8b40ddf24cbae06 : FAILURE in 17m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b730692435054667984449b40715a543 : SUCCESS in 6m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/da65cfe11aa24ceab12c1073e466bc45 : SUCCESS in 11m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e4e3bd830da47b2bd8654c5ed64bd56 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/289f35ef074445adbbebd2bbbee9163a : SUCCESS in 5m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/153e4044406840cdbb40ab4e4af7d3a4 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/69a990b7b0214835a833edc5f6a768b6 : FAILURE in 18m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/667f7a82eb654592854a78426d9664fe : SUCCESS in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1680cd70f13a461bb80aaf9d7b4327c4 : FAILURE in 19m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/17e331c2bb7648b39e29f9523aa2f847 : FAILURE in 25m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/555d5dd6fc58485fa971a0df69cf9e83 : FAILURE in 22m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a06caa7aa534a968bc0a5a6214c2bf4 : FAILURE in 29m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a22bc4ecfee44760ab9f407a7eb366a2 : FAILURE in 18m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/715a5240ccad4e43927065239d75fe2b : FAILURE in 16m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed286535ecb24e6ea4f6f2cc59274d99 : FAILURE in 21m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0deab3e81c745f384c917db9cf1e7b4 : FAILURE in 16m 36s","accounts_in_message":[],"_revision_number":14},{"id":"4ae5e98fc80f3c88961d2ff64b5c0b6deb0c773c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-31 10:45:36.000000000","message":"Patch Set 14:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"297168a2800b3c9986dae3cd629a5d948bf032d8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-31 10:55:44.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"6f62ca27ecf6e3e0ea147289c7b7f8957953b5ef","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-31 11:08:31.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"3f3d51c4930745c8364e6e8b0c92bd1665b2c7e6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-31 11:10:32.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/14/check/tempest-dsvm-full-bionic/dfbfaa7/ : FAILURE in 47m 01s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/14/check/devstack-plugin-ceph-dsvm-bionic/5850841/ : FAILURE in 48m 33s","accounts_in_message":[],"_revision_number":14},{"id":"21eaaffd9bef54a222fa80c35112503bb0525889","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-31 11:46:11.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"81863393b5ebadc8090d6af6c196c46560969305","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-31 13:00:44.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_XtremIO_FC/318fe5b/DellEMC_XtremIO_FC/783 : SUCCESS in 1h 35m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_XtremIO_iSCSI/c401516/DellEMC_XtremIO_iSCSI/749 : FAILURE in 16m 08s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":14},{"id":"29396597a3ff82277f3edf5c37c3130c7d295917","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-31 13:13:27.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_Unity_FC/630a505/DellEMC_Unity_FC/539 : SUCCESS in 1h 18m 33s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_Unity_iSCSI/194584a/DellEMC_Unity_iSCSI/514 : SUCCESS in 1h 40m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"ed928e8a9f75675d32acdd8118d53b0e4bb3c0bf","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-31 13:33:23.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"6230c29e8a6fef3ce45cb8fa1ef0df6fe17e1712","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-31 18:19:13.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/14/check/nimble-iscsi-driver-dsvm-volume/a15ca53 : SUCCESS in 3h 04m 32s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/14/check/nimble-fc-driver-dsvm-volume/2f23e53 : SUCCESS in 2h 51m 38s","accounts_in_message":[],"_revision_number":14},{"id":"31ec8153614fba891d5b64e5af43a87e66f10dcc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-01 04:45:44.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_PowerMAX_FC/a622124/DellEMC_PowerMAX_FC/566 : SUCCESS in 3h 56m 33s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_PowerMAX_iSCSI/1938aa1/DellEMC_PowerMAX_iSCSI/493 : FAILURE in 4h 31m 20s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":14},{"id":"9018039738ea3261175969f1046db673d12bc380","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-01 04:48:28.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"82578ab36d77407d16606aa5adc9e2d01ba743e9","author":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"date":"2020-09-01 08:56:54.000000000","message":"Patch Set 14:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":14},{"id":"2013ebae1acda94ef8379d6b9c88fe62d5f7755a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-01 09:37:10.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/14/check-not-vote/ext-cinder-zuul/924dea9 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":14},{"id":"4df91d189bfc530c5dce2c7da8929827440c3128","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-01 09:40:42.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"2d055e195b2d050b7edc6a1d64642b7c24d9122c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-01 20:11:02.000000000","message":"Patch Set 14:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-14/ns5_nfs : SUCCESS in 5h 8m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-14/ns5_iscsi : SUCCESS in 3h 4m 49s","accounts_in_message":[],"_revision_number":14},{"id":"6c98d36c2e0244f0afa672893fdd8e855cc4e617","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-02 14:14:22.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"82ec0a2daa1aa39127c20b2561ecf0975c28fff5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-02 14:19:41.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/14/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b5ff2ef : FAILURE in 4m 57s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/14/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/71d7943 : FAILURE in 4m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":14},{"id":"7412504d9842cc6f27c471792deeb13828a0c45d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-02 14:33:58.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/14/check-not-vote/ext-cinder-zuul/5760550 : FAILURE in 17m 18s","accounts_in_message":[],"_revision_number":14},{"id":"1e97f38aecd838be4ad4490882dcc9837e333476","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-02 14:50:14.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"2ab0c1ce8c3251ce8d6b2e4b1e88c3c84fc3bc6a","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-02 15:06:15.000000000","message":"Patch Set 14:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"45b294a72aa913b780438ce1855c6a1212f014d7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-02 15:12:53.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/14/iscsi/ : FAILURE in 25m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/14/smb/ : FAILURE in 58m 05s","accounts_in_message":[],"_revision_number":14},{"id":"62c0b3d3fa83bf69b93bd932241f0633a4dd5639","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-02 16:06:09.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"e19d7898d81b7c276032806604f03dc4a3ebacdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 17:10:18.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7f4169d559d1486ca83e9e88afba5985 : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9ac746b33ea41c4b538d1bc1bf80888 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d19fd624504fb890816a0234ece288 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90ec1ec228dd4d13a4a83b64d8ba1af0 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feea0513d4f0400988be33cab804baff : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f3009b558e468392c617e4bb460f88 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/e7f3abc323d04e5fa4383b1a03ca455b : SUCCESS in 55m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ea7c7f1a57b416abe6a5d52b49d25ae : SUCCESS in 1h 14m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fe2fa57477e48218cf84a48b56a9902 : SUCCESS in 8m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/907bc01de6c54f0382955862d77f3fe2 : SUCCESS in 13m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19f4714429b246e2808d911fbd0e1820 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae6301fab4b646c29a4ed492321b1da0 : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b43551f0143b4973827da08741cf1b68 : SUCCESS in 8m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/94fa51d2462f40ceb903e22f41d88db0 : SUCCESS in 40m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8ec9e3295e24ca2963b013b5ca763af : SUCCESS in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9707e37e9ef443acafcaabf3ef6eaac4 : SUCCESS in 1h 23m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b7d789fd235406db8cb3a7c4606ffa2 : FAILURE in 1h 45m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/21d9356b3e3744b597a49ef6ae562cf7 : SUCCESS in 1h 36m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/85fb5feb6fb7459ba032816d360b1821 : SUCCESS in 1h 19m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8692051bdd044e67ba33a719fbc6f23b : SUCCESS in 1h 14m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/23082e726ffd4c63af183f417217e2c5 : SUCCESS in 1h 30m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/713bbd96309a4c9ca56acabcc86ea245 : SUCCESS in 2h 03m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a7bd99f2138407ea36b070be675a904 : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":14},{"id":"4443f0acb333d7a05d55984ed39f2875e56c27ab","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-02 17:18:21.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/14/check/synology-dsm-cinder-iscsi/15898ab : FAILURE in 7m 29s","accounts_in_message":[],"_revision_number":14},{"id":"a89c3ce9b9e29c56a9210cdc8a587ddfa11abcbf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-02 17:34:32.000000000","message":"Patch Set 14:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-14 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":14},{"id":"de460182ebc6d76b94fdf51d8c992d538ff4aaf9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-02 17:57:07.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/14/upstream-solidfire-check/cinder-solidfire-iSCSI/4de0d40/ : FAILURE in 45m 39s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":14},{"id":"1de7a7fbc436eb66fbc3b2ce2afe1f408e29e5c4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-02 17:59:56.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/14/upstream-check/cinder-cDOT-iSCSI/07e112e/ : FAILURE in 29m 59s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/14/upstream-check/cinder-cDOT-iSCSI-IPv6/41f915d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/14/upstream-check/cinder-cDOT-NFS/ad0f3af/ : FAILURE in 30m 07s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/14/upstream-check/cinder-cDOT-NFS-IPv6/50aae7c/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/14/upstream-check/cinder-cDOT-FCP/94ab82b/ : FAILURE in 48m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"7f5ccbe6ae7c3f586252d3bf45c871a4795977cf","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-02 18:20:52.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"368600c3596fb690786292197fd15c42f1866d27","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-09-02 19:08:33.000000000","message":"Patch Set 14:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-14/Lenovo-iSCSI : SUCCESS in 43:42.44\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-14/Lenovo-FC : SUCCESS in 42:57.72","accounts_in_message":[],"_revision_number":14},{"id":"e7f8a6f54a6d0b0883f87d3b138004a08c6d19ea","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-09-02 19:10:54.000000000","message":"Patch Set 14:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/737707-14/HPMSA-iSCSI : SUCCESS in 45:28.40\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/737707-14/HPMSA-FC : SUCCESS in 42:51.73","accounts_in_message":[],"_revision_number":14},{"id":"0d0d481cf78e831520a5bab439e24703c934f9c8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-09-02 19:24:11.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_PowerStore_iSCSI/a52756d/DellEMC_PowerStore_iSCSI/186 : SUCCESS in 1h 03m 34s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_PowerStore_FC/92005ca/DellEMC_PowerStore_FC/187 : SUCCESS in 1h 07m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":14},{"id":"2697399f27583f722df1e7ecb06578cff011a1c7","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-09-02 19:25:15.000000000","message":"Patch Set 14:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-14/Seagate-iSCSI : SUCCESS in 49:54.11\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-14/Seagate-FC : SUCCESS in 48:48.24","accounts_in_message":[],"_revision_number":14},{"id":"98e7b45daefea2108a8900f96d9ae8edd805f2ad","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-09-02 20:19:54.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/07/737707/14/check/DellEMC_PowerFlex/97ba193/DellEMC_PowerFlex/452 : SUCCESS in 2h 07m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"abd23120d83cd6b47c5fd7288447eb62b6cfd6aa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-02 20:25:36.000000000","message":"Patch Set 14:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/14/check/tempest-dsvm-full-bionic/bc099ca/ : SUCCESS in 1h 42m 30s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/14/check/devstack-plugin-ceph-dsvm-bionic/de359cf/ : SUCCESS in 1h 40m 41s","accounts_in_message":[],"_revision_number":14},{"id":"d33e2bc2e206e3892991b7f83bcda687bd8396b4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-02 21:03:03.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"6cf2faa4dd8fbbb2eda9ea6c0e5df1004c57fc39","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-02 21:53:35.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_VNX_iSCSI/03cea5c/DellEMC_VNX_iSCSI/344 : SUCCESS in 2h 51m 45s\n- DellEMC_VNX_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_VNX_FC/6a5642e/DellEMC_VNX_FC/352 : SUCCESS in 1h 11m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":14},{"id":"996876126e2873e622c3db412d537e2c19da5f6c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-02 21:58:27.000000000","message":"Patch Set 14: Code-Review-1 Review-Priority+1\n\n(13 comments)\n\nThis is looking good.  A few things I\u0027d like you to look at noted inline.\n\nStill need to look at the tests, will do that after dinner.","accounts_in_message":[],"_revision_number":14},{"id":"1d406de2d9e5681809fc9db43e9eaba147e8bf0c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-09-02 22:04:12.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_SC_FC/506f27e/DellEMC_SC_FC/399 : SUCCESS in 1h 15m 38s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_SC_iSCSI/7271bce/DellEMC_SC_iSCSI/369 : FAILURE in 2h 09m 14s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":14},{"id":"20a060568eb86ca288565bc10d9a5e6464b039cd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-02 22:28:16.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"8060243aeca67c038663dd277ab2fa2ef26d1b97","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-03 01:00:19.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_XtremIO_FC/04f2c87/DellEMC_XtremIO_FC/806 : SUCCESS in 1h 36m 00s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_XtremIO_iSCSI/543a2d7/DellEMC_XtremIO_iSCSI/772 : SUCCESS in 1h 39m 17s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":14},{"id":"a1c1749e4cdf22e614ba088d8f3add74cea4ca20","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-03 01:40:30.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"c59641c6b0f030873157ff244fa359e09b9a3672","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-03 01:48:31.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_Unity_FC/b019b0b/DellEMC_Unity_FC/560 : SUCCESS in 1h 13m 57s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_Unity_iSCSI/1ec7371/DellEMC_Unity_iSCSI/533 : SUCCESS in 1h 43m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"9823f102fabfc3b2443969b3ca15fced7d5d97ad","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-03 02:24:00.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"b1818aafa85218cfdddaad67158cf1e9b0fc070d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-03 14:11:49.000000000","message":"Patch Set 14:\n\n(3 comments)\n\nTests look good, a few comments inline.  You seem to be light on negative tests, but the code coverage report looks really good.","accounts_in_message":[],"_revision_number":14},{"id":"eb5db60bae55548a100ac0cc521a441f1cba1668","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-03 17:13:08.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/14/check/DellEMC_PowerMAX_FC/9fece07/DellEMC_PowerMAX_FC/581 : SUCCESS in 4h 06m 55s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/14/check/DellEMC_PowerMAX_iSCSI/4f6e32a/DellEMC_PowerMAX_iSCSI/507 : SUCCESS in 3h 59m 26s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":14},{"id":"feb8616ed94d97a5064a1ca4d6e6fd03db10349b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-03 17:46:06.000000000","message":"Patch Set 14:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"34e925b250cf589277d585d08c7259384afe16e1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-03 18:19:28.000000000","message":"Patch Set 14:\n\n(1 comment)\n\nJust noticed that you\u0027ll need to update the spec (it has the responses as anonymous JSON objects/lists).  You made the right choice, this keeps the responses consistent with other resources.","accounts_in_message":[],"_revision_number":14},{"id":"2dac090016f39a6369932f053582413b169acbf7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-03 20:33:38.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/14/20200904030637/NEC-ISCSI : SUCCESS in 2h 14m 18s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/14/20200904030637/NEC-FC : SUCCESS in 2h 26m 59s","accounts_in_message":[],"_revision_number":14},{"id":"215c7866142eee1f1c84c418eb145d85ddbf6bfb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-04 12:01:58.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"faa44b902bb8a22240c0413f74e9a1ca3edba359","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-04 12:02:39.000000000","message":"Patch Set 15:\n\n(15 comments)","accounts_in_message":[],"_revision_number":15},{"id":"ad89a617d05a36c138c55b689934e24fcc4ae31f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-04 12:12:24.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/15/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/6d06515 : FAILURE in 5m 19s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/15/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/544e61a : FAILURE in 4m 44s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":15},{"id":"01ecfbffdc3e283565e11e3e0ab4f85ac4b7ec9d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-04 12:31:26.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest cinder-storpool-tempest : RETRY_LIMIT","accounts_in_message":[],"_revision_number":15},{"id":"8f40aa345e33999dd02047e46624bcae7ef4a990","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-04 12:37:31.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/38564244802684645dd35bf29e60a67713b59a3b/index.html : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"f90e8462ae6c0cc2d9e175c14c10d66960b9db63","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-04 12:42:28.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/340d42a/340d42a.html : POST_FAILURE in 7m 29s","accounts_in_message":[],"_revision_number":15},{"id":"1fd6cd7a7d8f548ab2e3f6bdf663ade4f10599ef","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-04 12:50:34.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/15/iscsi/ : FAILURE in 47m 57s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/15/smb/ : FAILURE in 47m 45s","accounts_in_message":[],"_revision_number":15},{"id":"bb1fbb033de687ac2213f654e47994a76e65df2c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-04 13:28:00.000000000","message":"Patch Set 15:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/15 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"76687fdfe7ec18248a21e4fdfe9caea7031ccf73","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-04 13:32:30.000000000","message":"Patch Set 15:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/15 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"daf6e627b1e74ee9a62229ca9e46901267990f14","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-04 14:01:17.000000000","message":"Patch Set 14:\n\n(1 comment)\n\nQuick comment about the 403 vs. 404 issue inline; haven\u0027t had time to look at the other changes yet.","accounts_in_message":[],"_revision_number":14},{"id":"909d046a13e74dea09d5a0a7ae3860828f264d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:17:09.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/769c908de3874b79b388e9e8e019463d : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53bcfec036964339951f55deb3b0bd33 : FAILURE in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b005e3f2a4d41ee94cd2b20c01b68d3 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e156a0cd57b84fa082502b1ec331ef6c : FAILURE in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d905b65112c4fdbb6672bd0c9c9d08f : FAILURE in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dffcdde6e7ed4daba8719cb75f6e01dc : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/836ae263cba44c31aef89ef9aaf02b0f : SUCCESS in 1h 01m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4dc8094b72d475e8a62a8dfc6eb434a : SUCCESS in 1h 08m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2c984c427984defa7d74349eca5d9a1 : SUCCESS in 8m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5eb549c0daa240939f294c71c2c5193e : FAILURE in 12m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f0211475e72c40fa832e4a330395b11b : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f27c95afd1e4df4b8af3f53b45782fc : SUCCESS in 7m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5349551091b34d439c05f181f468ccd7 : SUCCESS in 5m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dfd530fd8a1e4507bb5c7846b0a40bb4 : SUCCESS in 40m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc280cbcdcc5401c8aae0cb435b62c61 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1dcdbe4b5db047dcaae28b5ce9e7afea : SUCCESS in 1h 18m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3ea16b4de9c64f3a9e7243ed00214283 : FAILURE in 1h 09m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ecb9fe6f9634ce2802a027e5ee6cce4 : SUCCESS in 1h 22m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3106b4dc1897498b9dcce0bdeaca3bd6 : SUCCESS in 1h 08m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a91360ab4cc9492db919e12bad142192 : SUCCESS in 1h 20m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6c8f5fc83aec43e088d05e4d9571bef9 : SUCCESS in 1h 12m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d72c20229f246b1984ef7c059984afb : SUCCESS in 2h 05m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eff5450f8c9647a7851b79b7c463f38f : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":15},{"id":"7ad5b2935d9cdd4e7fd217891d5c343fc494d800","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-04 14:17:41.000000000","message":"Patch Set 15:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-15/ns5_nfs : SUCCESS in 2h 10m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-15/ns5_iscsi : SUCCESS in 2h 15m 24s","accounts_in_message":[],"_revision_number":15},{"id":"138f40e5e3c8791063422235ae27f29f656351b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-04 14:26:48.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/15/check/tempest-dsvm-full-bionic/dbc2527/ : SUCCESS in 1h 46m 52s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/15/check/devstack-plugin-ceph-dsvm-bionic/e89f8e4/ : SUCCESS in 1h 41m 30s","accounts_in_message":[],"_revision_number":15},{"id":"3db3755c18d9e899d16b8274991b2410f13da043","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-04 14:51:54.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/15/check/DellEMC_XtremIO_FC/ef5d86e/DellEMC_XtremIO_FC/None : EXCEPTION\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/15/check/DellEMC_XtremIO_iSCSI/7cd0e19/DellEMC_XtremIO_iSCSI/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"37eb597aaf3d67080685d29350d7e169d81a5732","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-04 15:09:27.000000000","message":"Patch Set 15:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/15 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"228673d802d0128064fb2d290e51c1e307c294c6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-04 15:25:54.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"1a955d24b3e951c2926832feafdfb80a45f5c172","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-04 15:28:16.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"0acf831c933d5adad178e3e68e0a1c1082591f77","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-04 15:31:09.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/16/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/29c43ed : FAILURE in 4m 54s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/16/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/9b00aeb : FAILURE in 4m 45s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":16},{"id":"3fcea4e2e52320e1ef4960e92c5224d34b12afff","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-04 16:11:46.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/16/iscsi/ : FAILURE in 45m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/16/smb/ : FAILURE in 45m 04s","accounts_in_message":[],"_revision_number":16},{"id":"dfac165ad6ea87d6dd02049433d85f5a664baa93","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-04 16:18:39.000000000","message":"Patch Set 16:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"9f5e73dfde9546c179e91bf6e31610feca9e029d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-04 16:33:00.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/16/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a23ec10/a23ec10.html : FAILURE in 1h 05m 43s","accounts_in_message":[],"_revision_number":16},{"id":"504fb333be8f88df66c0d17db9411fc38baa0cee","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-04 16:41:12.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/dbd8ef20bff571c71c3f64e4c2e3f2ae83b22556/index.html : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"915228b9276774a986f9a7e6211204a60bb34657","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-04 17:32:00.000000000","message":"Patch Set 16:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-16/ns5_nfs : SUCCESS in 2h 5m 44s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-16/ns5_iscsi : SUCCESS in 1h 51m 55s","accounts_in_message":[],"_revision_number":16},{"id":"a9f7f19a4e0a252f42821fe9a119d62a9d8d77a7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-04 17:51:24.000000000","message":"Patch Set 16:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"babf7387028a8311302c279e8db981c1c0ed738b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 18:41:15.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6d8e9302c2004a71a78ddedcfa06af23 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c98fa70fc91046f7949bcb39c2f40b57 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff1585aa4a84cc6ae5f53884efa50a3 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36a9e19f3b0a49c6b9e880eaec86089a : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3068aac0a77b42d49bcdf9a3721b84cb : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9196d9185041d0b2634e26938561fc : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/59cebacd612f4db3ab6438e791241cc9 : SUCCESS in 1h 09m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a797a8a72003419e88794af2967a1dbe : SUCCESS in 1h 14m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eebd306fc49465883ea33aedc703d0a : SUCCESS in 7m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/575c9fe9684d49dc8d772670645b31a1 : SUCCESS in 14m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/571a91bbc2d048f0ad3b94d07aece2f9 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b836950dc8424688abfd0874a84361d5 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37e9aa185f944da4afe6277f14f3cd5f : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bae4b815ed48427cbcf8a79062796344 : SUCCESS in 39m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98d895dc659a413d8b884f48366b9a0f : SUCCESS in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35694dbb22e8407da81c81b5fc6e8733 : SUCCESS in 1h 01m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/55126123a3724a44a585b0d05e7290e7 : FAILURE in 1h 09m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/059d22d5781e409fab91110e8cbe0f62 : SUCCESS in 1h 46m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9abbb155add547a9a5a563ad57149a62 : SUCCESS in 50m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3c76abf3191f4d76a9c392521cbafcb5 : SUCCESS in 1h 12m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3aa182ed971b43158bf3ef1a4060c924 : SUCCESS in 1h 17m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1078fd3b46ec4cbdb1fa64672472b3b0 : SUCCESS in 1h 46m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca6e5a89bee14ef9888118c76c645326 : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":16},{"id":"aa9d620b735894ded07308e87ef4601591dc5d79","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-04 19:00:27.000000000","message":"Patch Set 16:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/16/check/synology-dsm-cinder-iscsi/9cb0c41 : FAILURE in 7m 21s","accounts_in_message":[],"_revision_number":16},{"id":"b3da22431b55729f77987ca1063785e00a228d9e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-04 19:08:06.000000000","message":"Patch Set 16:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"e0520a8ee86761b72b23fe28867a613efbc464a4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-04 19:26:42.000000000","message":"Patch Set 16:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-16 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":16},{"id":"84219278aba42bd9633d14112a46efdeeba3f30a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-04 20:02:57.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/16/check/tempest-dsvm-full-bionic/8a804a7/ : SUCCESS in 1h 48m 53s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/16/check/devstack-plugin-ceph-dsvm-bionic/59ffdff/ : SUCCESS in 1h 44m 55s","accounts_in_message":[],"_revision_number":16},{"id":"8014d84dc47c69d4a911dd501a687a3d0525290d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-04 20:04:23.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/07/737707/16/check/DellEMC_VNX_iSCSI/b4ac5ba/DellEMC_VNX_iSCSI/None : EXCEPTION\n- DellEMC_VNX_FC http://publiclogs.emc.com/07/737707/16/check/DellEMC_VNX_FC/710cc2d/DellEMC_VNX_FC/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":16},{"id":"6d6ae318a055065164597f8c3e6511e75409dcb5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-04 20:05:20.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/16/upstream-check/cinder-cDOT-iSCSI/3aa3d39/ : SUCCESS in 36m 34s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/16/upstream-check/cinder-cDOT-iSCSI-IPv6/8d0f58b/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/16/upstream-check/cinder-cDOT-NFS/8b196f6/ : SUCCESS in 35m 47s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/16/upstream-check/cinder-cDOT-NFS-IPv6/045e7db/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/16/upstream-check/cinder-cDOT-FCP/3bb5e21/ : FAILURE in 1h 23m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16},{"id":"4d0ac1365cb99122b6423069abbd362921e6b389","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-04 20:47:24.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/16/upstream-solidfire-check/cinder-solidfire-iSCSI/04dbd47/ : SUCCESS in 2h 05m 00s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":16},{"id":"00c7ad926fc286aadfa048dee52a4e0695b67fe5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-04 21:53:43.000000000","message":"Patch Set 16:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"5c8d2e720ccb1eb55795388be39b2d35004b42ec","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-09-05 01:45:12.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/07/737707/16/check/DellEMC_PowerFlex/c776af9/DellEMC_PowerFlex/468 : FAILURE in 1h 57m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"ff4a802041fe49ba215ebf6dcb6f07cfce8dca12","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-05 03:23:00.000000000","message":"Patch Set 16:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"37984c2d082ccf8707825521cb7e6699d755ad75","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-09-05 03:25:10.000000000","message":"Patch Set 16:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-16/Seagate-iSCSI : SUCCESS in 41:15.46\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-16/Seagate-FC : SUCCESS in 42:03.09","accounts_in_message":[],"_revision_number":16},{"id":"453d36563f61e97951e2a26631fac2468307b077","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-09-05 03:44:06.000000000","message":"Patch Set 16:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/737707-16/HPMSA-iSCSI : SUCCESS in 49:09.21\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/737707-16/HPMSA-FC : SUCCESS in 50:06.14","accounts_in_message":[],"_revision_number":16},{"id":"5405bea537838393ece2b35d2e6f67eea97e2e60","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-09-05 04:30:53.000000000","message":"Patch Set 16:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-16/Lenovo-iSCSI : SUCCESS in 46:11.58\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-16/Lenovo-FC : SUCCESS in 45:28.16","accounts_in_message":[],"_revision_number":16},{"id":"e4a48a0168bfcd9f9603112d975d5b11dc1ba38e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-09-05 06:01:56.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/07/737707/16/check/DellEMC_SC_FC/fc3f8af/DellEMC_SC_FC/416 : FAILURE in 1h 21m 20s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/07/737707/16/check/DellEMC_SC_iSCSI/5b82a15/DellEMC_SC_iSCSI/385 : FAILURE in 2h 37m 16s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":16},{"id":"07527907eaa269081d004cf15955b6791baf327c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-05 06:37:24.000000000","message":"Patch Set 16:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"f3f6383b9794067471904a17007bd64e6c77b96e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-06 00:20:55.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/16/check/DellEMC_PowerMAX_FC/22c3116/DellEMC_PowerMAX_FC/595 : SUCCESS in 4h 10m 09s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/16/check/DellEMC_PowerMAX_iSCSI/d158ab1/DellEMC_PowerMAX_iSCSI/520 : FAILURE in 4h 09m 21s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":16},{"id":"27ffb6051e2355f48a608f9cf888ffd508ea1106","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-06 00:54:09.000000000","message":"Patch Set 16:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"d03daced41e854276f4c61ea6bab2952dbddd7ec","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-06 08:31:02.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/16/20200906150513/NEC-ISCSI : SUCCESS in 2h 12m 36s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/16/20200906150513/NEC-FC : SUCCESS in 2h 25m 47s","accounts_in_message":[],"_revision_number":16},{"id":"e678bf1240686c54610ab965ea0d00ca447c6b2d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2020-09-06 16:38:08.000000000","message":"Patch Set 14:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/07/737707/14/fc : SUCCESS in 0h 50m 59s","accounts_in_message":[],"_revision_number":14},{"id":"e98c8253d581230dc65dc713b8ed485470e2d7e8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-07 06:19:26.000000000","message":"Topic changed from 737707 to bp/multiple-default-volume-types","accounts_in_message":[],"_revision_number":16},{"id":"b947307ba8f554ebe20c5376649bd39ea0f9a2e1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-07 14:16:21.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"92839a288ecbd2600f02e4b60707ae3a106dfe85","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-07 14:21:37.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/17/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/618c22b : FAILURE in 4m 54s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/17/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f4a652c : FAILURE in 4m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":17},{"id":"d44448d7fce3479aa002acf34df7302608340a4c","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-07 14:31:28.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/e044393a536656e9939ee5a512ec4a42de8321f3/index.html : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"47294dee4a5cc4ddc56ed3e3f614f5905740a017","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-07 14:49:06.000000000","message":"Patch Set 17:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"d008bce7c787f2d6dee5e750a7083890f8ef67c5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-07 15:03:43.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/17/iscsi/ : FAILURE in 44m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/17/smb/ : FAILURE in 46m 51s","accounts_in_message":[],"_revision_number":17},{"id":"3a419ca0d4adfa1a300ac551f3bf2ad70477bdf7","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-07 15:07:37.000000000","message":"Patch Set 17:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"195badd06ded67ebc219e0bc1eb270a98cc2f62a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-07 15:11:43.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/17/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/929ef78/929ef78.html : FAILURE in 52m 53s","accounts_in_message":[],"_revision_number":17},{"id":"e7a545af4773f87a75aa69503893d091d42a97f0","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-07 15:25:22.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/17/check/dsvm-tempest-zadara-driver/d199137 : SUCCESS in 1h 06m 09s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"57149d0ce80676ef0630a64af8daa85807657275","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-07 15:36:48.000000000","message":"Patch Set 17:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"5622de26245a07d1f798580d899b8cd4ecd33488","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-07 15:37:51.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/17/check/cinder-storpool-tempest/bfc38fc/ : SUCCESS in 1h 09m 51s","accounts_in_message":[],"_revision_number":17},{"id":"eeffa7b582d4832aae102d35d0797dd499cfcdc5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-07 16:19:52.000000000","message":"Patch Set 17:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-17/ns5_nfs : SUCCESS in 1h 49m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-17/ns5_iscsi : SUCCESS in 2h 3m 6s","accounts_in_message":[],"_revision_number":17},{"id":"dd065fb56bf0637f0b57652f36456aaf087c49bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 16:20:17.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dd11f7e2a6ce4cf8b8d05cac317e2711 : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/587479c0bb6949148fae0c029a8f63c1 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d80b7d122ddc47af9afb46bce08b5da5 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42cda57cb1b84bf4b6f3b635f6241ce4 : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ac2b2463f004f3598965a7b255bedaf : FAILURE in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cc5698115c140478c10f86074a97787 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/849c38a2e629436d932d18d13152dcff : SUCCESS in 1h 02m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0564053e7d7e4e74a50c116263abebdf : SUCCESS in 54m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90e9356e9745492c8d28e5a61e4560bf : SUCCESS in 5m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e0d78f31675e4103894db936ea7d9a1a : SUCCESS in 9m 02s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/00da8393d1a64c5baf0c940b83723924 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/498cbe7a70464e00a027b1b92eb40b53 : SUCCESS in 8m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55bd8bb7d8cd4f6b9a19c677b959ca6c : SUCCESS in 5m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/216392dadaf54ad19886b6c901beac19 : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75f62f2e8ed44207905134c8e0c8a7a8 : SUCCESS in 8m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/97700e1d5bf74f698bd26a150a2ed753 : SUCCESS in 1h 19m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/438147a1f28348739a6ec02ad0bffe2d : FAILURE in 53m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/90f4b6618769458288c9c8cd28a9a7ea : FAILURE in 58m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dd79e35b215748d3991196855fa75cc1 : SUCCESS in 1h 19m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/53ad231fd6e84ceca7c5f667a610171a : FAILURE in 1h 01m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/591c0995ee2f4891b35ae1674c831179 : SUCCESS in 1h 31m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/876f29eb1a274760af2d7acc4e89c394 : SUCCESS in 2h 00m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c04eee8582347c79abd96026de5765d : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":17},{"id":"babd68e0ee67b79eb4657a58c5a5862c27bf9ad1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-07 16:36:02.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"0ad8727950ac024f8138ebbd69d1c84860a07387","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-07 16:41:12.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/17/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ae069e7 : FAILURE in 4m 52s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/17/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/03a9b84 : FAILURE in 4m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":17},{"id":"0e787e67371e65b39530687eb1afd58beae448bb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-07 17:14:54.000000000","message":"Patch Set 17:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/17/check/tempest-dsvm-full-bionic/59cd589/ : SUCCESS in 1h 25m 11s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/17/check/devstack-plugin-ceph-dsvm-bionic/e01879e/ : SUCCESS in 1h 46m 47s","accounts_in_message":[],"_revision_number":17},{"id":"bc7dc9c13f72968f51cd2b150b414889ea28a50b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-07 17:23:11.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/17/iscsi/ : FAILURE in 46m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/17/smb/ : FAILURE in 46m 34s","accounts_in_message":[],"_revision_number":17},{"id":"747ce3a352aa7916b35b0be872ee845b1e666263","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-07 17:27:47.000000000","message":"Patch Set 17:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"41c6a229c37f6268a740110d95ffdc008ff6c08d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-07 17:27:50.000000000","message":"Patch Set 17:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"8ed572ad3b1d10d833ea9aa2d869668806800b6b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-07 17:34:59.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/17/check/DellEMC_Unity_FC/f295932/DellEMC_Unity_FC/581 : FAILURE in 1h 18m 01s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/17/check/DellEMC_Unity_iSCSI/31a42be/DellEMC_Unity_iSCSI/554 : FAILURE in 1h 42m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"a8b923884f8a6c50fd607372c1674434af1d8346","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-07 17:44:20.000000000","message":"Patch Set 17: Code-Review-1\n\n(28 comments)\n\nRegardless of the number of comments, the feature works great (except for the problem with the default policy).\n \n-1: Please update the docs in doc/source/cli/cli-manage-volumes.rst where it talks about the \"default_volume_type\".  It should include some reference to the project defaults and how users can get the effective default with \"cinder type-default\".","accounts_in_message":[],"_revision_number":17},{"id":"4f5d46b0b0093eefcd58b772abf6d0d505fb8929","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-07 17:58:37.000000000","message":"Patch Set 17:\n\nBTW, I know we agreed in the spec to return volume_type_id when listing, but after playing for it for a while I think that wasn\u0027t a good idea, it\u0027s pretty inconvenient.\n\nI DON\u0027T WANT to change the current implementation, but it could be a future improvement to be able to resolve the volume_type_id to name, and also optionally the project_id to project name.","accounts_in_message":[],"_revision_number":17},{"id":"8d8569836f378a73083212644623e1a2fc158fd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 18:48:44.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c52ed4393e6a4a81b786a47ee8d48253 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9997ecebedee435b8165d43023eab1b3 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d67671b3b84dfc933b2686bc783dd5 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3841df4b21924147b522e73db1c94eca : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/757236ec5a81433497a8528836a66c43 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3f4bae2a73451d86b2ce545526a921 : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/197b33cb9d104b59b83d7f007d2ce5cf : SUCCESS in 1h 00m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e6cdc5c225b4f52b1d9a7095724e0ed : SUCCESS in 1h 25m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88602cce02e344568c84530597903999 : SUCCESS in 8m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1bc1ad55ba4847469308bf3ee95add6b : SUCCESS in 11m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09431d90b14d4251b2e6ff2f808b5ca1 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/16b7639761954aae991b1b46ead4f3b8 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7e5de681c7d40bb9b3864d771711a53 : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/711c5a604e024335971adf737714d97b : SUCCESS in 38m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b8810e2c8dc47729c7ebeeb89f9faa5 : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d2dd5edf4a0144e5ae9e5e7dfb7b5bcb : SUCCESS in 1h 08m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1fb0f7436c8f4a20ba9ef4f3897c30ec : FAILURE in 1h 05m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec460c8167ef4b96b63396685d9e6e33 : FAILURE in 59m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2dbb5587a0304beeb9a289cac925f77e : SUCCESS in 1h 12m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b8b25798b1d8466aa0b4a0e4e9a56d96 : FAILURE in 1h 12m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/13d4ae36b4284ed3a9c82f8e10a09c41 : SUCCESS in 1h 25m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d10c10cc8b3450e957160a36a999459 : SUCCESS in 2h 05m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45e7ebe6796246faa6c6767a0be62d5f : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":17},{"id":"0393af757926b7e4f385b74a9d8fc4d8c94deded","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-07 18:50:51.000000000","message":"Patch Set 17:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"99a74327d7f57aabaf2f14ca8c936f0fd8185cf4","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-07 19:03:00.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/17/check/dsvm-tempest-zadara-driver/8fee68e : SUCCESS in 1h 05m 27s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f0f5e66de84deebc91c393154dc5fd7064ddb832","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-07 19:17:24.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/17/check/cinder-storpool-tempest/f0e3ffe/ : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":17},{"id":"1c2313b41c3648ec34ad377ac250f34b57f1a7b0","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-07 21:07:42.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/17/20200908034017/NEC-FC : SUCCESS in 2h 13m 18s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/17/20200908034017/NEC-ISCSI : SUCCESS in 2h 27m 23s","accounts_in_message":[],"_revision_number":17},{"id":"18cbe4545a91c52d08156f1d7800942f72aa682a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-08 10:38:42.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"114a1c3b1c65c1674de372c4343bad721315fd1d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-08 10:39:32.000000000","message":"Patch Set 18:\n\n(26 comments)\n\nThanks for the review Gorka\n\n \u003e (28 comments)\n \u003e \n \u003e Regardless of the number of comments, the feature works great\n \u003e (except for the problem with the default policy).\n \u003e \n \u003e -1: Please update the docs in doc/source/cli/cli-manage-volumes.rst\n \u003e where it talks about the \"default_volume_type\".  It should include\n \u003e some reference to the project defaults and how users can get the\n \u003e effective default with \"cinder type-default\".\nDone","accounts_in_message":[],"_revision_number":18},{"id":"9392781866c0210d9e0e5ccc79b47dfac84dc31a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-08 10:42:14.000000000","message":"Patch Set 18:\n\n\u003e BTW, I know we agreed in the spec to return volume_type_id when\n \u003e listing, but after playing for it for a while I think that wasn\u0027t a\n \u003e good idea, it\u0027s pretty inconvenient.\n \u003e \n \u003e I DON\u0027T WANT to change the current implementation, but it could be\n \u003e a future improvement to be able to resolve the volume_type_id to\n \u003e name, and also optionally the project_id to project name.\n\nThat is a good suggestion. Since that won\u0027t change the number of parameters in request/response, it won\u0027t require a new mv bump.\nwe can discuss the idea with team and change later.","accounts_in_message":[],"_revision_number":18},{"id":"4def76c4aed2da04d999d9f443aceedc233a6221","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-08 10:45:23.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/18/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f04e7ad : FAILURE in 5m 07s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/18/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/0af1756 : FAILURE in 6m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":18},{"id":"c78f88af1d8ddd1253963c51b908fb34f4c81ec6","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-08 10:54:12.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/c5a547c560eeb329e9c20d14f438e5e57ffa7ff7/index.html : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"b6dbdd366153f886da902671603e8cf6144a7e3d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-08 11:31:07.000000000","message":"Patch Set 18:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"950ad1ad80b30ac9b78076d38a8b45d84c348e3b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-08 11:34:05.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/18/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4394f03/4394f03.html : FAILURE in 53m 02s","accounts_in_message":[],"_revision_number":18},{"id":"d09ea083f135954abeb0e280bcd31c5aff7bbca3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-08 11:35:52.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/18/iscsi/ : FAILURE in 46m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/18/smb/ : FAILURE in 56m 39s","accounts_in_message":[],"_revision_number":18},{"id":"96723c6e81407684a87e163f5e4f70bf5909e20d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-08 11:51:48.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"37527c2d29119b6353028a463735d68f72e28f00","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-08 11:57:35.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/19/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/1b02ef0 : FAILURE in 5m 16s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/19/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/aba5eb4 : FAILURE in 5m 25s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":19},{"id":"8a97a1ea10607482cfd665c858a41d84290f1499","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-08 12:05:57.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/c0a1b6efce83b8532220b7b468f3176afba7bcaa/index.html : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"06d1e56dba0d1de35d2fcebbf14fce5f682b8766","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 12:06:02.000000000","message":"Patch Set 18:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"ac6ded701f8cebbadd6d4bcd205b89e1b434af0d","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-08 12:21:42.000000000","message":"Patch Set 19:\n\n(12 comments)","accounts_in_message":[],"_revision_number":19},{"id":"059649e200be7fa07a5fe695941af9b09269dd6a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-08 12:22:55.000000000","message":"Patch Set 19: Code-Review-1\n\nSee patchset #18 for comments on why the downvote.","accounts_in_message":[],"_revision_number":19},{"id":"98bff06a1ec0c05616575d6262eae0ff853c055d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-08 12:30:22.000000000","message":"Patch Set 19:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/19 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"5ee847a86dcb7c12e9e430db6f12636a0db41a86","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-08 12:37:12.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"dd0d6b2790d7f5281f8aaddbdde48e474fc930b6","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-08 12:37:28.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/19/iscsi/ : FAILURE in 45m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/19/smb/ : FAILURE in 45m 09s","accounts_in_message":[],"_revision_number":19},{"id":"5760b09043971eb39598f58ae28d61008707eb86","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-08 12:43:28.000000000","message":"Patch Set 18:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-18/ns5_nfs : SUCCESS in 2h 4m 21s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-18/ns5_iscsi : SUCCESS in 1h 46m 46s","accounts_in_message":[],"_revision_number":18},{"id":"353b37ff496f620b0ed4bf9aa6c00c21115e8a0a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-08 12:46:07.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"2f0e098afa3bc37b8d764b466a7d85ecfdca3c76","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-08 12:47:44.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/19/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/b4b7138/b4b7138.html : FAILURE in 53m 29s","accounts_in_message":[],"_revision_number":19},{"id":"bca1b923a503ead7238a54c2371080cfc5d3ba0c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 12:47:53.000000000","message":"Patch Set 18:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"eabbbafc24884faa9c41beba314494e4354296a4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 13:24:31.000000000","message":"Patch Set 19:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/19 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"261eb003ce43377357bbf169c3a341298ec24041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:57:45.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8632bd1827814bccb8bd936e1f5685d5 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e1ca392cb134c27b5dc4e2b35be9c65 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4728ca3d73b84577b44b4cfc7c53efcb : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/817242484688498bba34a49e4b1b78ba : FAILURE in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbecadfd56df4d97931a289be878070d : FAILURE in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f94417c7d94ff6a3d5fff029926634 : SUCCESS in 11m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/ce0a164b4fbb453ea5c80013d483b430 : SUCCESS in 59m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9acccec84a5c4a2a83d536eaaccbb13a : SUCCESS in 1h 04m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc466af8d3df4e5cac4376640f33357e : SUCCESS in 7m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/56007dbb6cb647eabb4d9e879a7ce9ca : FAILURE in 9m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9e18bf5a05aa4c41bd45e443c96b4be6 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b95b077bc9e4f3f88f296640be86e06 : SUCCESS in 6m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a116c55ede454770973f25fcdcd833e8 : SUCCESS in 8m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/08dd2673e97d4bf391cc709f7003ab6c : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7af5139a327d4af19ce9923b93372f57 : SUCCESS in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b4e2512cbcad44efbd3e26fb0b1f91ab : SUCCESS in 57m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aa2945b1aff34df7ba119cbf366582b3 : FAILURE in 1h 09m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f3e9aa863f94640a15e4ae58fc8ffee : FAILURE in 1h 29m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d3ccd37d0fd04fc7b85540bba4087774 : SUCCESS in 1h 06m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f2247c268bce44dd9c72dfb6a2e7c0c1 : FAILURE in 1h 13m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d2146a373b184d7cab801e379e058200 : SUCCESS in 1h 06m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d29e744fddbd40169d22b63091e01667 : SUCCESS in 2h 03m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c07d8f9ab4b4fb1b6ada7c0fe54f5c3 : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":19},{"id":"543079fb83c859ffdfff7e90898a6b9a74e75916","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-08 13:59:14.000000000","message":"Patch Set 19:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-19/ns5_nfs : SUCCESS in 2h 5m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-19/ns5_iscsi : SUCCESS in 2h 7m 0s","accounts_in_message":[],"_revision_number":19},{"id":"9f7980a5f32bc4019bf85fb29713120e41fa876a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 14:34:51.000000000","message":"Patch Set 19:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/19 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"fc0ed1fc7745981a7b529e3e327c588b541659b6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-08 14:37:03.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/19/check/cinder-storpool-tempest/56cedaa/ : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":19},{"id":"52ce277a9d56639a924d7ce8b19e8d3e447f1d46","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-08 14:40:08.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"393be7833bff26c51a4936b4457a1a709dfe7341","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-08 14:45:26.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/20/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/906a3fa : FAILURE in 4m 52s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/20/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6f3de09 : FAILURE in 4m 57s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":20},{"id":"89661fb1b3fa82fb7651298e55d62acf6f6e5684","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-08 14:55:29.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/d786963e93865051a503eed26ed597021f9dd95f/index.html : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"e4ae3cd16a74f2b1018ddf649560ff5d2cdfe950","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-08 15:17:33.000000000","message":"Patch Set 20: Code-Review-1\n\n(3 comments)\n\nOnly one minor thing to go!!","accounts_in_message":[],"_revision_number":20},{"id":"7bcd8d7f117f75867863a60ee25195e9abe19f58","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-08 15:31:48.000000000","message":"Patch Set 20:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"239979ee9b6281236db646841cd44af527ef775c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-08 15:35:31.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/20/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0e07046/0e07046.html : FAILURE in 53m 50s","accounts_in_message":[],"_revision_number":20},{"id":"28da363623f7c8d0e8e22a20007b0533e5ef461d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-08 15:37:48.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/20/iscsi/ : FAILURE in 45m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/20/smb/ : FAILURE in 57m 10s","accounts_in_message":[],"_revision_number":20},{"id":"3e7aa04048bd796961e2d73d3381535ac2d07238","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-09-08 15:54:04.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"fa933bc39db4faf3035c67a2dfe7327d312da8c3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-08 16:33:25.000000000","message":"Patch Set 20:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-20/ns5_nfs : SUCCESS in 1h 50m 5s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-20/ns5_iscsi : SUCCESS in 1h 52m 51s","accounts_in_message":[],"_revision_number":20},{"id":"8aa5530880b2a0c8b25ab4dd808e5aae732c943f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-08 16:34:12.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"2194d70406717c8b362cb77016a0c557af2e94b6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-08 16:35:16.000000000","message":"Patch Set 21:\n\n(13 comments)","accounts_in_message":[],"_revision_number":21},{"id":"873da13bb5b4c15858548888bc6f96dd0935d974","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-08 16:39:22.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/21/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e42bd20 : FAILURE in 4m 48s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/21/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/44e038a : FAILURE in 4m 43s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":21},{"id":"10fb0d2497b56e5f339fe28fa932bcbee857f7f6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 16:43:39.000000000","message":"Patch Set 20:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"5ae0355be76187a8d6fb79ace5b3083ee4a36037","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-08 16:50:30.000000000","message":"Patch Set 21: Code-Review+1\n\n(2 comments)\n\nCode looks good, but I\u0027m still worried about not being able to make requests with the default policy.\n\nNot even manually requesting what should be a system scoped token, and then using curl.","accounts_in_message":[],"_revision_number":21},{"id":"47415bb39727b75088e25b72ca2ecea32ea06fff","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-08 16:54:04.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/c84b9113a38ccf7a95cb9565420b9f87ec1fa2e5/index.html : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"5990def856e0bf568386e1a53536ba0a4574bc7a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-08 17:00:43.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"f2a369a69a4ece85a6a5e691d4f25250fb5d5c96","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-08 17:06:50.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)\n\nI was finally able to get the token and make it work with the default policy.","accounts_in_message":[],"_revision_number":21},{"id":"9d88a81514eae2156974c7276faefa5af78e97b0","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-08 17:19:39.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/21/iscsi/ : FAILURE in 44m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/21/smb/ : FAILURE in 44m 46s","accounts_in_message":[],"_revision_number":21},{"id":"d5da8b066dffee53784c543e92924b91b1c6e09e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 17:21:25.000000000","message":"Patch Set 20:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"655db35eb8070753a5c49cd28bef344a68c1238b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-08 17:25:41.000000000","message":"Patch Set 21:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"cd9c48700a9ed27a04d5aa29a68c0c6cb7f48717","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-08 17:34:12.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/21/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e6c9ef1/e6c9ef1.html : FAILURE in 54m 04s","accounts_in_message":[],"_revision_number":21},{"id":"ef233e7f908c23ed11fb64c814a30e67ade02cab","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-09-08 17:38:39.000000000","message":"Patch Set 21:\n\n(1 comment)\n\nI think this might still need some work to properly consume system-scope. It looks like there is still at least one case where cinder is assuming project-scoped tokens.","accounts_in_message":[],"_revision_number":21},{"id":"ee3fe620bb0bee35b605e033dfe4f9b6f8adf96d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-09-08 17:41:02.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"df11a05611e7e127e8b6ca6b43714c9f3bf73d80","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-08 18:36:56.000000000","message":"Patch Set 21:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-21/ns5_nfs : SUCCESS in 2h 2m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-21/ns5_iscsi : SUCCESS in 1h 47m 46s","accounts_in_message":[],"_revision_number":21},{"id":"7e22b1842594f7cf853f555d0f6370ea2e29c39a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-08 18:41:28.000000000","message":"Patch Set 21:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/21/check-not-vote/ext-cinder-zuul/88c9aac : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":21},{"id":"03e214798421729024b3e2adf59f5a4b8dcb36df","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 18:43:27.000000000","message":"Patch Set 21:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"b0d99c2d77ad9e4694d7fcc56ef80098edbd4b32","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 19:18:28.000000000","message":"Patch Set 21:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"1da213f7e5e7371625184587c35729d87a2b5f23","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-08 19:34:04.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/21/check/cinder-storpool-tempest/3be5fb5/ : SUCCESS in 1h 13m 14s","accounts_in_message":[],"_revision_number":21},{"id":"e1573064e4bc971fc6809a48c4cc39043afddbea","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-08 19:39:39.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/21/check/dsvm-tempest-zadara-driver/fe16aab : SUCCESS in 1h 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"011c24369efc4ee848597f3a4480b12fbe609df5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-08 21:07:00.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/21/check/DellEMC_XtremIO_FC/c2f2785/DellEMC_XtremIO_FC/830 : SUCCESS in 1h 35m 06s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/21/check/DellEMC_XtremIO_iSCSI/1acc9b3/DellEMC_XtremIO_iSCSI/795 : FAILURE in 1h 38m 34s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":21},{"id":"e92e3ed1908c5454115c76c6170233824604b5c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:27:05.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fb02e5ff05ed4a52af865632d101de6d : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11725f75607245b8a9e8a05b81175379 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9ee9c83b66428b8dcad46a12a947ac : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0348bb6807f640f2bbd8ba808f4353ea : FAILURE in 11m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5496cec04724583bb25647a4d92b6e6 : FAILURE in 13m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df903dca58844f27a9d3c1a2211cc9f2 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/2a65333e51534267adcbe0fc82f1cfec : SUCCESS in 1h 02m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1487030910f43b68565e67442f71d12 : SUCCESS in 1h 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2db9c0b6d9344dd0945cef4db0a331b5 : SUCCESS in 8m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5bf8865e3bcf4e4896d5ac64513ab0eb : FAILURE in 11m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f4a54df120754df894ba913fc6e7d5a0 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45c8d657a75243cda017f815ab989793 : SUCCESS in 9m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ed3b438ecc840db9c0ef60177222a5c : SUCCESS in 7m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/520d4aa48a2c48e5901963dc12d7dd6f : SUCCESS in 52m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9409b1c2692a4fcfb422bad7d5987b6b : SUCCESS in 9m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/371eb2404f0f4448892ac1a8eb74a142 : SUCCESS in 1h 02m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ee054cda8274585a376171f8fc26beb : FAILURE in 1h 18m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c771d95a4ba4287a18a247a7f714f41 : FAILURE in 1h 29m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/98600a42c10747a29636d0eb3ee10a0a : SUCCESS in 1h 12m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bb615fea623e46e99880891f85e38ebf : FAILURE in 1h 17m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/64a23991dcdc48a1ba165b51203bf5b1 : SUCCESS in 1h 25m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37b72ce4e2f24f35ba33faa08dc3fcb4 : SUCCESS in 2h 06m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27cf19a429e74fcea9eae67c0c75f093 : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":21},{"id":"e0e1ca989dac88260f4d01efff9b930bf5092c2d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 22:50:27.000000000","message":"Patch Set 21:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"3fbd754c609d507e1b0382812c5505492705cfbd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-08 23:13:32.000000000","message":"Patch Set 21:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/21/check/tempest-dsvm-full-bionic/dda013b/ : SUCCESS in 1h 49m 16s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/21/check/devstack-plugin-ceph-dsvm-bionic/f2e4e81/ : SUCCESS in 1h 46m 00s","accounts_in_message":[],"_revision_number":21},{"id":"a5f46383da0eaca9b2ee190bfffb9abcc3a2d20e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-08 23:46:51.000000000","message":"Patch Set 21:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"e8bfb4280cabe7a714b6fcee7ede308adc520cc8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-09 00:10:03.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/21/check/DellEMC_Unity_FC/8870a23/DellEMC_Unity_FC/595 : FAILURE in 2h 18m 41s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/21/check/DellEMC_Unity_iSCSI/4d0bc85/DellEMC_Unity_iSCSI/568 : FAILURE in 1h 43m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"fd295a60243a6d735243cb6a727746d85a96e98b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-09 00:26:26.000000000","message":"Patch Set 21:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"cfb43ad0af9cf4bf6dad16a8f6dfebe473b73a0b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-09 02:15:04.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/21/20200909085036/NEC-ISCSI : SUCCESS in 2h 10m 54s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/21/20200909085036/NEC-FC : SUCCESS in 2h 24m 26s","accounts_in_message":[],"_revision_number":21},{"id":"43e1ea80671160483ffffa903f07ec165948aec5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-09 14:09:58.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"42d1bc3bb44eacddddde3c6dd36bfb64751f8b59","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-09 14:17:23.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/21/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3ba2e07 : FAILURE in 4m 54s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/21/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/deb0199 : FAILURE in 4m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":21},{"id":"ba2e53bbc1a46602561d6a6ba020a5d4c44de365","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-09 15:00:32.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/21/iscsi/ : FAILURE in 49m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/21/smb/ : FAILURE in 50m 08s","accounts_in_message":[],"_revision_number":21},{"id":"4f573ca5d4306c3573959e804d697aa4ffc4ed35","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-09 15:01:52.000000000","message":"Patch Set 21:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"ba5d6125d0cf2ed22c454927941bb2ce881f4f83","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-09 15:15:42.000000000","message":"Patch Set 21:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"33014358e992dff5cb9acb0cba65956ae4fdd229","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-09 15:36:48.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"31c50766d26cce6e5f61bcb81aa9421a2cfe8231","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-09 15:42:36.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/22/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e0f3311 : FAILURE in 5m 25s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/22/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/91f8167 : FAILURE in 4m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":22},{"id":"cf6c1bd5f8086b85d5995546c7ca5eef5ad9f9bf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-09 15:46:25.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/22/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1b875c7/1b875c7.html : FAILURE in 7m 40s","accounts_in_message":[],"_revision_number":22},{"id":"f9d764abaf45efd8a8744c5d01ef6c8dd416a2b5","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-09 15:52:21.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/1d37f45ceda0e426366ce893a6ee01ffa1b967f6/index.html : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"41c5698bf7f9cdf015a598edc481ea5e5613badf","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-09 16:25:58.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/22/iscsi/ : FAILURE in 46m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/22/smb/ : FAILURE in 48m 39s","accounts_in_message":[],"_revision_number":22},{"id":"d9929200b4d9c6a64b1d5c158692b83e44828148","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-09 16:29:37.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/22/check/dsvm-tempest-zadara-driver/34ea2f4 : FAILURE in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d0c0285ceff2a07bca52df76fb47ab48a87214b4","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-09 17:06:32.000000000","message":"Patch Set 22:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/22 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"6bde78b05558b2b440c00adb88ccb1f271abf27b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-09 17:12:27.000000000","message":"Patch Set 22:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/22 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"baa0e0927699cfc27536fe033160d05f2dfc76c6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-09 17:43:44.000000000","message":"Patch Set 22:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-22/ns5_nfs : SUCCESS in 2h 6m 36s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-22/ns5_iscsi : SUCCESS in 1h 54m 6s","accounts_in_message":[],"_revision_number":22},{"id":"27feb310d331b4420a704d8a5584e2a72c679a7d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-09 18:05:24.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/22/check/cinder-storpool-tempest/e24f759/ : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":22},{"id":"92f082c78cf0ba7a7286cc0ce071201791644b2e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-09 18:21:26.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/22/check/DellEMC_XtremIO_FC/bf0a4e5/DellEMC_XtremIO_FC/840 : SUCCESS in 1h 32m 10s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/22/check/DellEMC_XtremIO_iSCSI/cf70c6b/DellEMC_XtremIO_iSCSI/804 : SUCCESS in 1h 38m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":22},{"id":"6c3d0cc3dea1de13ca36bd457074d5753734be7b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-09 19:02:57.000000000","message":"Patch Set 22:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/22 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"25e36ffb6e0822e430391bbac033a20bf71b0252","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-09 19:46:12.000000000","message":"Patch Set 22:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/22/check-not-vote/ext-cinder-zuul/3222047 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":22},{"id":"f542cd2d809f419717853b66392a518cbba267e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 20:05:14.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/698cab471849470a96219e0dd071d15a : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/128efdf6b60a4184a8febd3db9ae5a9d : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aaf5c225cba48ce981f659c459b2dfd : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c1d697faadd44098a2dd559679ebe16 : SUCCESS in 9m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fe3dc9c3f3740d098d04a84fbfa80be : SUCCESS in 16m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98059fedafb49d588f20702d373c499 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/f9fa0df8afe540dd93b4425ffac3d037 : SUCCESS in 59m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce21de6d4a2f4633aa8ee183a96d06b1 : SUCCESS in 1h 16m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03dd937b89ec404184195abf26d7a2f1 : SUCCESS in 7m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b1a41d1adbdd4a9684ce156e752ba3e3 : SUCCESS in 12m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea2842e45515478ca973edaa6683c4d4 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5af8578184a945f7b3a09abe21b7bc73 : SUCCESS in 6m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8e3b1707519b41afa565c7f6fd50f026 : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/341342ccfecd40c0985221fb04811e5d : SUCCESS in 42m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/227acd02ec3c4d3da1cf164be61394b5 : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/37007c67cbf348abbcd4168b2da8c388 : SUCCESS in 1h 06m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/492d4ee36382460bb9fef73532c2232f : FAILURE in 1h 19m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2e2f13f4de54c81b07da061b5d0886d : SUCCESS in 1h 30m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e8931839593459ca82fcb80288c9f7b : SUCCESS in 1h 11m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/919ad46375584c4393da872a8ab0202f : SUCCESS in 1h 19m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/46712bbdd7a54204b7e0433461362746 : SUCCESS in 2h 59m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9313a17d2f14e5aaaefed23042da30d : SUCCESS in 2h 10m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c929bc83c7a4f0ca38a45b5090f3df6 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":22},{"id":"019b115125acfdf185cba1c057c67217523ab9d4","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-09 20:09:30.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/22/20200910024639/NEC-ISCSI : SUCCESS in 2h 20m 04s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/22/20200910024639/NEC-FC : SUCCESS in 2h 22m 48s","accounts_in_message":[],"_revision_number":22},{"id":"eb6be249a9f1e12ebf58596d175e90448b2dbf17","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-09 20:13:15.000000000","message":"Patch Set 22:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/22/check/synology-dsm-cinder-iscsi/ac1ce0b : FAILURE in 7m 28s","accounts_in_message":[],"_revision_number":22},{"id":"b56ea1752337a7b3b09e87bf1a84fe35782b11a2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-09 20:24:12.000000000","message":"Patch Set 22:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/22/check/tempest-dsvm-full-bionic/49ff906/ : SUCCESS in 2h 02m 38s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/22/check/devstack-plugin-ceph-dsvm-bionic/8bba9e4/ : SUCCESS in 1h 37m 50s","accounts_in_message":[],"_revision_number":22},{"id":"efc7a29f9f602fda97a13d0e1570c8093d8a4b56","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-09 20:25:23.000000000","message":"Patch Set 22:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-22 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":22},{"id":"f151900e715005086bb26159bd65a5cdb8ea0792","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-09 21:12:25.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/22/upstream-solidfire-check/cinder-solidfire-iSCSI/8edb60b/ : SUCCESS in 1h 05m 56s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":22},{"id":"1d9878a4c208f2ad557ee8d4ae5c19e3a9d7459f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-09-09 21:25:24.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/07/737707/22/check/DellEMC_SC_FC/c9d626e/DellEMC_SC_FC/428 : FAILURE in 1h 13m 10s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/07/737707/22/check/DellEMC_SC_iSCSI/0a074c0/DellEMC_SC_iSCSI/397 : FAILURE in 1h 19m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":22},{"id":"9a6701777bd54988ccfb092b485c36efffc500d3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-09 21:39:47.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/22/upstream-check/cinder-cDOT-iSCSI/6185bcc/ : SUCCESS in 1h 33m 39s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/22/upstream-check/cinder-cDOT-iSCSI-IPv6/c9a7b2c/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/22/upstream-check/cinder-cDOT-NFS/91a5cd0/ : SUCCESS in 38m 09s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/22/upstream-check/cinder-cDOT-NFS-IPv6/9a97cd9/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/22/upstream-check/cinder-cDOT-FCP/8718ead/ : SUCCESS in 58m 28s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":22},{"id":"e219ee3d818533aa1ab90b6d452e5af219d3ef8e","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-09-09 22:11:25.000000000","message":"Patch Set 22:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-22/Seagate-iSCSI : SUCCESS in 46:25.77\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-22/Seagate-FC : SUCCESS in 46:39.40","accounts_in_message":[],"_revision_number":22},{"id":"f9c9506124c15f1448362ee4204003b66b04ed56","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-09-09 22:15:36.000000000","message":"Patch Set 22:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-22/Lenovo-iSCSI : FAILURE in 52:10.60\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-22/Lenovo-FC : SUCCESS in 45:23.03","accounts_in_message":[],"_revision_number":22},{"id":"95ad992aaf6f686bf4688dc08760a1c6c12bbb9f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-09-09 22:22:49.000000000","message":"Patch Set 22:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/737707-22/HPMSA-iSCSI : FAILURE in 54:32.80\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/737707-22/HPMSA-FC : SUCCESS in 50:34.00","accounts_in_message":[],"_revision_number":22},{"id":"17796a3f07449d7e9f709bff2273279053953d76","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-10 00:33:30.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/22/check/DellEMC_Unity_FC/427d395/DellEMC_Unity_FC/604 : FAILURE in 2h 14m 10s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/22/check/DellEMC_Unity_iSCSI/15dfe1f/DellEMC_Unity_iSCSI/577 : FAILURE in 1h 55m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"67b631eb6171812e63d1bcedc857f22017dbcb9d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-10 00:34:25.000000000","message":"Patch Set 22:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/22 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"202a6ebb1549b43663bae9dcf8388dfa5d43ca5f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-09-10 00:50:21.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/07/737707/22/check/DellEMC_PowerFlex/2186544/DellEMC_PowerFlex/480 : SUCCESS in 2h 08m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":22},{"id":"1cb9030213fc9dc942d6780f0cb3fd67b6a36eac","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-10 02:03:16.000000000","message":"Patch Set 22:\n\n(1 comment)\n\nQuestion inline about the DB table.","accounts_in_message":[],"_revision_number":22},{"id":"5421d20559f3886c7c3780df8b6826243384961a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-10 02:12:50.000000000","message":"Patch Set 22: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"446362305bcaa184842b28f62b7755688d7a890f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-10 06:44:39.000000000","message":"Patch Set 22:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/22 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"34f092d4a49b1cfac821988576d3d73e2babce15","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-10 06:48:45.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4ac9bb67ab659bff97e36d8efb1e389b7bafccf9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-10 07:33:01.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/22/check/DellEMC_XtremIO_FC/ea6d406/DellEMC_XtremIO_FC/848 : SUCCESS in 1h 34m 22s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/22/check/DellEMC_XtremIO_iSCSI/5b2197d/DellEMC_XtremIO_iSCSI/812 : SUCCESS in 1h 37m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":22},{"id":"24eca78c4253b0b421b6bb5a09b67eac85adf22f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-10 09:07:15.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/07/737707/22/check/DellEMC_VNX_iSCSI/8a161fb/DellEMC_VNX_iSCSI/364 : SUCCESS in 2h 48m 13s\n- DellEMC_VNX_FC http://publiclogs.emc.com/07/737707/22/check/DellEMC_VNX_FC/4a1caf9/DellEMC_VNX_FC/372 : SUCCESS in 1h 04m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":22},{"id":"be2d21b0096093225b50a79d52b6b45b8faa77ec","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-10 11:36:49.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"ab15ae395ec21c8d8beb1bb7fe69b08a1655e872","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-10 11:42:31.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/23/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/10b9eb6 : FAILURE in 5m 19s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/23/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/bda5166 : FAILURE in 5m 13s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":23},{"id":"fb61f8eac7635c8c1de0d076b197a4be698be205","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-10 12:00:54.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"8fe164efc7ed39a0192b92bbf5ca86031fa2eae6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-10 12:17:10.000000000","message":"Patch Set 23:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/23/check-not-vote/ext-cinder-zuul/589622a : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":23},{"id":"d8031706608b5d98ced2b05baacb694cee0c7c05","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-10 12:27:14.000000000","message":"Patch Set 23:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/23/iscsi/ : FAILURE in 48m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/23/smb/ : FAILURE in 49m 36s","accounts_in_message":[],"_revision_number":23},{"id":"4cf361b74f756ee450fcce156ba55c8e701e0ed8","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-10 12:28:56.000000000","message":"Patch Set 23:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"b9621136251a37903c2ae74de2dfe9eb71d7ec55","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-10 12:34:21.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/23/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a72fc38/a72fc38.html : FAILURE in 53m 08s","accounts_in_message":[],"_revision_number":23},{"id":"0c85b2170c6e3d5b96fa8f838f7703854469b49f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-10 13:24:08.000000000","message":"Patch Set 23:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/23/check/tempest-dsvm-full-bionic/c77c169/ : SUCCESS in 1h 37m 55s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/23/check/devstack-plugin-ceph-dsvm-bionic/469da57/ : SUCCESS in 1h 30m 06s","accounts_in_message":[],"_revision_number":23},{"id":"2c65cdb994f597e702cea9f145279915ad99d2cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 13:37:29.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4ffb0f219462475e98e56a9e9a8db310 : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03084f441da94a8f936d28c38390fb3f : FAILURE in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e631d7bd464ed0891a293ade9acac6 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2487b49b76874c7e8bf3b7ef8b2e3795 : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d88c065e1164a0ba58e7b4d380aa9a0 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5689b70d823646a08e0eb2fba9b42cea : SUCCESS in 22m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/66d9b19a8dc34d8e95f2a68325034994 : SUCCESS in 1h 03m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/388a714f05f5490787b2102a62918fb1 : SUCCESS in 1h 01m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/590a6af9e1e4450292900fc5b9bd295c : SUCCESS in 6m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d36097439ad437d92082e39fa1ac58e : SUCCESS in 8m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/451a6fbc2839483bb46fd8248c9616d1 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d33632649ee4dbd8516c060c288ab37 : SUCCESS in 6m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/227cc6d08dfc4d148bf6daa7099ad015 : SUCCESS in 4m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f6773f5d1f274992becd57fdc36d49e8 : SUCCESS in 43m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2557652344a4b0291ecb3d9ec93e26e : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ce5aede4d654f75a5b5c5ee321f4fbc : SUCCESS in 1h 06m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/47aafb96d778439794072f9b56db1e72 : FAILURE in 1h 17m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52c170b8b2e647e489c69bea3771f747 : SUCCESS in 1h 57m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf10f3ba8c79466fbfe1c18d160534fd : SUCCESS in 1h 07m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fdfd2ee7c30b4b2c976680cfaacce324 : FAILURE in 1h 18m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/905fcfba20114d38ac3bcefc7a82389f : SUCCESS in 1h 21m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16739edd745f42d89f20ac38c878c480 : SUCCESS in 1h 27m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec8ddc04d06f409081158db3d6a0c468 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":23},{"id":"26aaaefb10e710f7d4af09a2219c813bc0030f67","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-10 13:41:37.000000000","message":"Patch Set 23:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-23/ns5_nfs : SUCCESS in 1h 48m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-23/ns5_iscsi : SUCCESS in 2h 4m 30s","accounts_in_message":[],"_revision_number":23},{"id":"30c9c43216001c94ba66b013ecf60c7ca515a474","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-10 13:44:16.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"bf4cd8f5567a74649fb5e04afdebb1776b3160a0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-10 13:49:26.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/23/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/251b332 : FAILURE in 4m 48s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/23/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/5a48b7c : FAILURE in 4m 51s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":23},{"id":"fbd6d824c82923256a3367a81505185961c89e0a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-10 14:17:09.000000000","message":"Patch Set 22:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/22 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"46546ea3d928c6145dba31675009b5c2f451372a","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-10 14:29:45.000000000","message":"Patch Set 23:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/5a370f2cb9a39a9636b70f6b4a21e72cc331abb0/index.html : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"2e7495d726950772a0f6237ac6978dcbcae1d502","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-10 14:35:26.000000000","message":"Patch Set 23:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"089fb0b1b627e7bce8108c2156f43e5aa1720692","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-10 14:43:16.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/23/check/DellEMC_Unity_FC/17abefe/DellEMC_Unity_FC/608 : FAILURE in 1h 23m 12s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/23/check/DellEMC_Unity_iSCSI/588c206/DellEMC_Unity_iSCSI/584 : SUCCESS in 1h 41m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"5ffc98335e80839107d399a64edef091887f90f1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-10 15:57:03.000000000","message":"Patch Set 23:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/23/iscsi/ : FAILURE in 56m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/23/smb/ : FAILURE in 52m 03s","accounts_in_message":[],"_revision_number":23},{"id":"c1fed71cd146d3d730c36e17693898b13c087402","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-10 16:35:03.000000000","message":"Patch Set 23:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"68481c2988e4836048cfe6290cd1c5cd72595264","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-10 17:12:08.000000000","message":"Patch Set 23:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"42501b87885e784b01f5e3ece5c94d43deee5cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 17:27:33.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/615eb2f2b1664da191c44a90cfe3cc3d : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19b24fc9b28345b68a2f7cfa048572f4 : FAILURE in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a86726f53654b33bba86fb7bdcb1c07 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5432d7d715e4cb2800558d88727ead3 : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/452781b4a6a34f1b8b4be4e3470a000c : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0df3c3a99674334b0a3c6a3a82bc55f : SUCCESS in 30m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/fd416599c91844af91352f428038aec4 : SUCCESS in 1h 03m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/346f61ab543a488680f102cced42ced5 : SUCCESS in 1h 18m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/363d4131e359416d93c1e6149d88c840 : SUCCESS in 7m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5f3705d881f84ffc897af7a0feec5e1d : SUCCESS in 14m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cf7203ef67642f792ca070938f8361a : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/16e3a4938c804f1a8c8bd88c452af03b : SUCCESS in 6m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a11fa53ffe44e779bdde8be026036c1 : SUCCESS in 7m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b88cc36c28cd42eeb39bd14adbaa7115 : SUCCESS in 42m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6fe26dab803470b94e0b04e885ec5d4 : SUCCESS in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6cb8cacdf3c648a683cdd93e2361db7a : SUCCESS in 1h 02m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7716e06b54fa46138b134fbf4c0ce411 : FAILURE in 1h 23m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7306992bf5724336a4928d457a5e14c8 : SUCCESS in 1h 19m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9279aae3e44948a3bc7ba70214b64fdd : SUCCESS in 1h 37m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/52f10b16372b401581f36dde1d0696e9 : SUCCESS in 1h 21m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/54d6aa7bd8504d1eb7bfa2365cdb1ea4 : SUCCESS in 1h 25m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79c788a91eaf452fb59bec5c3fb8bffa : SUCCESS in 2h 13m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/653d1826589d46bc9abd210e6634c7d8 : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":23},{"id":"067c326546629b458d0574a9a41c530e0139b13e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-10 18:34:10.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/23/check/cinder-storpool-tempest/3849e21/ : SUCCESS in 1h 10m 33s","accounts_in_message":[],"_revision_number":23},{"id":"20a88dfed8f4ce930f04031a1e1c2123f3328908","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-10 20:10:21.000000000","message":"Patch Set 23:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/23/check/tempest-dsvm-full-bionic/e7ed006/ : SUCCESS in 1h 46m 52s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/23/check/devstack-plugin-ceph-dsvm-bionic/db2a521/ : SUCCESS in 1h 38m 15s","accounts_in_message":[],"_revision_number":23},{"id":"a68d73834cbbf0efb2cf93f2f3ee1fe4bb893fec","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-10 20:54:14.000000000","message":"Patch Set 23:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/23/check-not-vote/ext-cinder-zuul/fda0532 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":23},{"id":"c9e56e08487717a11d50e04430943667a90e3d86","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-10 21:41:14.000000000","message":"Patch Set 23:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"5159f6021875ac2fe1a41cdf5463caba6a0de2ee","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-10 23:09:58.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"a118c8a3934ad973d7be0e580b648d5df92c1f44","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-10 23:15:22.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/07/737707/23/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/84d85a8 : FAILURE in 5m 01s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/07/737707/23/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b7307ff : FAILURE in 4m 48s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":23},{"id":"fcda62e7d7a789a2f76aa54b403d9766923077d4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-10 23:57:48.000000000","message":"Patch Set 23:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/23/iscsi/ : FAILURE in 47m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/23/smb/ : FAILURE in 46m 38s","accounts_in_message":[],"_revision_number":23},{"id":"5d539c4d994ee812f20f75d6fbdd4a050a2a7f50","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-11 00:01:15.000000000","message":"Patch Set 23:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"dc8942c123e755404e436c9967c85f78d8efa0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 01:22:12.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/78f9337101994684bb05b19f9263c7a0 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f136a22bb506468fbb842bec5bfcc038 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f366e534edd246d889bf719de9f11df4 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8ee41be8da468883b09b5ae2c1f853 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eba8a4ce039144b78d7ba4d41b24c649 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a09528653b4f50891d1560aead91d2 : SUCCESS in 20m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c22e1a32bf274e79987a8138c538e153 : SUCCESS in 1h 02m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21f283f840264869b0bab865f9c496d2 : SUCCESS in 58m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a54a21cf689f49478e4d5dd7ebde966c : SUCCESS in 8m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d8cd299142c9420bb18d43a2303028ab : SUCCESS in 12m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/270e12afb8794c03b079577593408fe3 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cdbc5f30b7d499cbbd8c0d0c06c506d : SUCCESS in 6m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8886720f3f54323904bc4bea28aeba1 : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/16ee425b1d0a4b33b145bb8a7ac6b846 : SUCCESS in 41m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f54012c1cac4442799b22113634738b6 : SUCCESS in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a4f34f351cd94b6592c35fa1d6636e7e : SUCCESS in 1h 02m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2d25581ce428466b82c9b6342a20b116 : FAILURE in 1h 21m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bbdbd745d19747f5909c8fa4a69be6e2 : SUCCESS in 1h 29m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5c0f7e7b30448178dd3d9e26b617f43 : SUCCESS in 55m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0ec8bc55ad4f481094efae018d271dd1 : SUCCESS in 1h 08m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2bf86c0be6a4482eae1e6fa75a36f4d9 : SUCCESS in 1h 08m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/797317198ab24485a32d40be7ffb2bb4 : SUCCESS in 2h 06m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bd5e117a28744e3a82269ab9508b05f : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":23},{"id":"b1a1c8a70bef840e02664208525feb97b5ac103b","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-11 01:31:09.000000000","message":"Patch Set 23:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/23/check/synology-dsm-cinder-iscsi/2986822 : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":23},{"id":"9ff28b72e0a8a82bcb35e2fa711e9e1c7225ce12","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-11 01:42:29.000000000","message":"Patch Set 23:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-23 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":23},{"id":"bb8fb175ccea2fd01ee9e8b8d584517731d5cc51","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-11 01:47:21.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/23/check/dsvm-tempest-zadara-driver/e7fc9b4 : SUCCESS in 1h 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"58f346c94916e300eb057c7b7c02490e0b5ab53e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-11 02:01:39.000000000","message":"Patch Set 23:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/23/check-not-vote/ext-cinder-zuul/44b8c28 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":23},{"id":"467b6a598f0bbee969ff3cae048acfa93edeb757","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-11 02:08:02.000000000","message":"Patch Set 23:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"9e17fe97e1dd0eead28c939e78a0c56f5a78b81c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-11 02:31:51.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/23/upstream-check/cinder-cDOT-iSCSI/fbf96e9/ : SUCCESS in 33m 47s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/23/upstream-check/cinder-cDOT-iSCSI-IPv6/5c11b3e/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/23/upstream-check/cinder-cDOT-NFS/2ba642a/ : SUCCESS in 35m 50s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/23/upstream-check/cinder-cDOT-NFS-IPv6/6d9a195/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/23/upstream-check/cinder-cDOT-FCP/c8bf838/ : SUCCESS in 1h 08m 36s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"a5a0cea1d72f5ebb1b763671aa5c472d26605b12","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-11 02:40:16.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/23/check/DellEMC_XtremIO_FC/1df71d2/DellEMC_XtremIO_FC/857 : SUCCESS in 1h 34m 46s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/23/check/DellEMC_XtremIO_iSCSI/7096182/DellEMC_XtremIO_iSCSI/821 : SUCCESS in 1h 40m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":23},{"id":"3c49bedfabaab83733c9ed46b4e541f370250a99","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-09-11 02:44:34.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/07/737707/23/check/DellEMC_SC_FC/2a23576/DellEMC_SC_FC/432 : FAILURE in 1h 14m 03s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/07/737707/23/check/DellEMC_SC_iSCSI/5cdf392/DellEMC_SC_iSCSI/401 : FAILURE in 1h 22m 04s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":23},{"id":"5acba68edaa740ee785b52eaa349da0ebe74f0ed","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-11 03:09:16.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/23/check/cinder-storpool-tempest/76562ba/ : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":23},{"id":"26de4236d6818e00c0e129489d26008d95adfdbe","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-09-11 03:29:55.000000000","message":"Patch Set 23:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-23/Seagate-iSCSI : SUCCESS in 41:21.58\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-23/Seagate-FC : SUCCESS in 39:11.76","accounts_in_message":[],"_revision_number":23},{"id":"defdd3c2d169c82871fe2cfc519c557f3adc71bd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-11 03:35:11.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/23/upstream-solidfire-check/cinder-solidfire-iSCSI/2fd380f/ : SUCCESS in 2h 11m 43s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":23},{"id":"0d1e2c6132429764448c9b49d7d0fed0c14e3e12","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-11 03:55:36.000000000","message":"Patch Set 23:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/23/check/tempest-dsvm-full-bionic/26c265d/ : SUCCESS in 1h 28m 26s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/23/check/devstack-plugin-ceph-dsvm-bionic/c4ef68c/ : SUCCESS in 2h 05m 15s","accounts_in_message":[],"_revision_number":23},{"id":"62bc012666e3c73876ec21239ac72bfefd7d78f4","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-09-11 04:12:34.000000000","message":"Patch Set 23:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/737707-23/HPMSA-iSCSI : SUCCESS in 59:09.96\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/737707-23/HPMSA-FC : SUCCESS in 51:26.77","accounts_in_message":[],"_revision_number":23},{"id":"caf235517fa3eb773625f2842621486f542b73e1","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-09-11 04:14:55.000000000","message":"Patch Set 23:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-23/Lenovo-iSCSI : SUCCESS in 1:00:40\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-23/Lenovo-FC : SUCCESS in 52:35.24","accounts_in_message":[],"_revision_number":23},{"id":"b1dfacbbfee906b80542c00472aaaa44a1496d76","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-11 05:29:13.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/07/737707/23/check/DellEMC_VNX_iSCSI/99842d6/DellEMC_VNX_iSCSI/367 : SUCCESS in 2h 53m 38s\n- DellEMC_VNX_FC http://publiclogs.emc.com/07/737707/23/check/DellEMC_VNX_FC/c028f59/DellEMC_VNX_FC/376 : SUCCESS in 1h 13m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":23},{"id":"ac7b9fcff860fa68d9ca37bcd5aed98aee1a97ce","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-11 07:53:16.000000000","message":"Patch Set 23:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"0f0b60f56b5916be502e84827854e3f3bdbb27a1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-11 08:26:29.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/23/check/DellEMC_Unity_FC/4946a7a/DellEMC_Unity_FC/620 : SUCCESS in 1h 24m 47s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/23/check/DellEMC_Unity_iSCSI/cdadf69/DellEMC_Unity_iSCSI/594 : SUCCESS in 1h 45m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"8072ebd977f740d51faae20afe10079e00490be9","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-11 10:12:03.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/23/20200911165523/NEC-FC : SUCCESS in 2h 14m 39s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/23/20200911165523/NEC-ISCSI : SUCCESS in 2h 16m 38s","accounts_in_message":[],"_revision_number":23},{"id":"230630a567a7fb7f95fd2d5c9f19011066bd6522","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-11 11:34:04.000000000","message":"Patch Set 23:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"85c9d8f70a166ca609a38ceb9e3d6f789f07f888","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-12 15:16:07.000000000","message":"Patch Set 23:\n\n(1 comment)\n\nComment inline.","accounts_in_message":[],"_revision_number":23},{"id":"b419cb5a30676131c8ad26a9a0eaf415780bc72c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-12 15:33:53.000000000","message":"Patch Set 23:\n\nI\u0027ll summarize real quickly my concern about the DB design and the introduction of an \u0027is_active\u0027 column.  Intuitively, \u0027is_active\u0027 \u003d\u003d True is already indicated by \u0027deleted\u0027 \u003d\u003d False.  The problem it\u0027s being introduced to solve here is to be able to have a uniqueness constraint on project_id (since there\u0027s only supposed to be one of these active for any project_id); we need to account for deleted rows.  But this table is different from the volumes or snapshots tables in that rows don\u0027t represent entities, they represent a relation between a project and a volume_type.  So whereas keeping deleted volumes around in the DB may make sense because they are records of entities that existed at one time, this default_types table is an entirely different thing.  My thought is that this table really just needs a uniqueness constraint on project_id (as a backup for a bug in the code), and we handle things like this: for set/update (same operation in this API), make a model_query with read_deleted\u003d\u0027yes\u0027 and if there\u0027s a row present, use it (setting all columns appropriately); otherwise create it (setting all columns appropriately).  I think all the other calls will work correctly based on the value of \u0027deleted\u0027.\n\nI just want to make sure we really need this extra layer of complexity before we add it.","accounts_in_message":[],"_revision_number":23},{"id":"eda88ee08bee177dd7f6e54c2a715d2ad5620a87","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-14 11:09:27.000000000","message":"Patch Set 23:\n\n(3 comments)\n\n@Brian, the reason for that DB nuisance is to keep records of previous default types, just like we keep them for old volume types (which don\u0027t represent entities either).\n\nI think that, besides the value of having a consistent behavior with the soft deletes in the DB, it may also have value when investigating some situations, as we can know what the default type was at some point in time.\n\nIf we decide we don\u0027t want to keep old records, I would prefer we did hard deletes instead of soft deletes with \"resurrections\".","accounts_in_message":[],"_revision_number":23},{"id":"8e026cd95db7d9bc4375e7d5f2534532422baf8b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-14 14:14:09.000000000","message":"Patch Set 23:\n\n\u003e If we decide we don\u0027t want to keep old records, I would prefer we did hard deletes instead of soft deletes with \"resurrections\".\n\nI agree with this.  We have a precedent with the workers table: https://opendev.org/openstack/cinder/src/commit/e4103150748a07e00c101a058d5a835bc833c95c/cinder/db/api.py#L1777-L1779  This should really simplify the DB code for the default_types table, and it\u0027s appropriate because all it records is a relation between a project and a volume_type that either holds or doesn\u0027t.","accounts_in_message":[],"_revision_number":23},{"id":"27f9a559e96fed71d40735e69242c1b4f17d1329","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-14 20:08:33.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"febe2f2698fe75f3101806a475280c739cb8f631","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-14 20:08:58.000000000","message":"Patch Set 24:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/07/737707/24/check/fujitsu-eternus-dx-iscsi/2ed1580 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/07/737707/24/check/fujitsu-eternus-dx-fc/75b5e65 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":24},{"id":"9d24024452dc7eb995ca1c2eb3469ed036297913","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-14 20:46:23.000000000","message":"Patch Set 24:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/06294f4906fa721bac23d86aa29aafa7c58fdcd4/index.html : SUCCESS","accounts_in_message":[],"_revision_number":24},{"id":"85f026f6041569a679f4a7c97973a56d1dd58307","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-14 20:56:10.000000000","message":"Patch Set 24:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/24/iscsi/ : FAILURE in 46m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/24/smb/ : FAILURE in 47m 02s","accounts_in_message":[],"_revision_number":24},{"id":"12a54f6dcb7f632fdb3bc2c764a2819b96cca710","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-14 21:00:07.000000000","message":"Patch Set 24:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"fb965c124d50fe7a42a5306778125715c7783581","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-14 21:02:53.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/24/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/18cf250/18cf250.html : FAILURE in 52m 31s","accounts_in_message":[],"_revision_number":24},{"id":"eeac78c705ead8a5494af2e848ff3f6961b887a8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-14 21:44:06.000000000","message":"Patch Set 24:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/24/check-not-vote/ext-cinder-zuul/eb0463d : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":24},{"id":"6a59e4532f0dbdd96d43522a6e64dc3dfab57968","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-14 21:58:25.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/24/check/tempest-dsvm-full-bionic/bd7e2fe/ : SUCCESS in 1h 31m 21s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/24/check/devstack-plugin-ceph-dsvm-bionic/a4f36d1/ : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":24},{"id":"bc3dfe65940ad95ed659feaf4f55df01d1c7863c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 22:09:31.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9fa99277692b4e07884c64a3072b6001 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/522f6952d7904c60ac5915461fb00a5a : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cab2a4dee974e108fe3aa0c32c41c1d : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/337aa8d5e0b0417b9cb26829982f0b72 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976c6f363ded41af99da0dd4e8c5ba91 : SUCCESS in 21m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/abb0a437e80540dbaccb83ca9b7d05d4 : SUCCESS in 56m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/138bd0420b8747dca2545da5ad608e26 : SUCCESS in 1h 02m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/526613a053994204b9b3d59916b3e4d5 : SUCCESS in 7m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/331efbf970224693acc2284a1132823a : SUCCESS in 12m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da6608ac3b334d85a21a297b31d3702c : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/757d40b68ad949ed9bfee53f4b772458 : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bcbcbdf1d4d4e6c87f4ee0266b5461f : SUCCESS in 9m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dbd2396b922e4c499d9a0746ef3bcacc : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e052e80dec744dc48758d2ac5cbb1a57 : SUCCESS in 41m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a03815dee8f4d60a430e8374246fd1d : FAILURE in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9423756c4ed49929fece66dfcde38d4 : SUCCESS in 1h 01m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/898116029644433c8d894ab59a103f95 : FAILURE in 1h 15m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b876298a57c944f4894e6729ebfd641f : SUCCESS in 1h 39m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4073bbd8ae204219b1ecdd1702a1d611 : SUCCESS in 1h 12m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3cd317bb31a34e5eb6cfd16c56d80686 : SUCCESS in 1h 20m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a71be372c5614fca8685ee9c48b5fde4 : FAILURE in 16m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ff8eb555d684ae694cc2e3348037ca8 : SUCCESS in 1h 59m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d712caf57b694e7da658b1a691bcabdc : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":24},{"id":"ad4922dda071e3bd7e719585b0040e252499f172","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-14 22:12:26.000000000","message":"Patch Set 24:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-24/ns5_nfs : SUCCESS in 2h 3m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-24/ns5_iscsi : SUCCESS in 1h 51m 12s","accounts_in_message":[],"_revision_number":24},{"id":"c877ec4cae85a932be7a9a7d869069f6f32c6dc5","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-14 22:18:49.000000000","message":"Patch Set 24:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/24/check/synology-dsm-cinder-iscsi/fe23196 : FAILURE in 8m 44s","accounts_in_message":[],"_revision_number":24},{"id":"cfafa8b81cc0a0d4496ab749b2c8f3a7e334de4d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-14 22:29:39.000000000","message":"Patch Set 24:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-24 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":24},{"id":"c85264883fc62810301625002862d78f0395ed6e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-14 23:06:16.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/24/upstream-check/cinder-cDOT-iSCSI/7523ab7/ : SUCCESS in 39m 59s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/24/upstream-check/cinder-cDOT-iSCSI-IPv6/37e3244/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/24/upstream-check/cinder-cDOT-NFS/82220b4/ : SUCCESS in 38m 34s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/24/upstream-check/cinder-cDOT-NFS-IPv6/60cc8a8/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/24/upstream-check/cinder-cDOT-FCP/6736376/ : SUCCESS in 55m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":24},{"id":"c341dbe37a126b77178d8e6984e12291b79e9280","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-14 23:14:26.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/24/upstream-solidfire-check/cinder-solidfire-iSCSI/10e7883/ : SUCCESS in 1h 03m 39s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":24},{"id":"522c9a97bdb17e691535c3f37c80faba3f0bb8aa","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-09-14 23:29:25.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/07/737707/24/check/DellEMC_PowerStore_iSCSI/26def98/DellEMC_PowerStore_iSCSI/203 : SUCCESS in 1h 10m 58s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/07/737707/24/check/DellEMC_PowerStore_FC/eff87bd/DellEMC_PowerStore_FC/204 : SUCCESS in 1h 09m 08s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":24},{"id":"24cf54c50389831a9ab7eb718922ecd611129f67","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-14 23:50:47.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/24/check/cinder-storpool-tempest/ecf18d9/ : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":24},{"id":"20943e6c520c2ffbbf4d3f91b97887a01ebf192d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 00:26:52.000000000","message":"Patch Set 24:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"bd047bd7fb2e566c76c089c3c776c751d70b8fcc","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 00:27:53.000000000","message":"Patch Set 24:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"2a026dc0c03a386f39ad0eeafc4c95df90874245","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-15 01:59:19.000000000","message":"Patch Set 24:\n\n(1 comment)\n\nFrom this patch and the ML, it doesn\u0027t look like there\u0027s a lot of interest in the database question.  I think we should keep it simple, then, and hard-delete from the default_types table.","accounts_in_message":[],"_revision_number":24},{"id":"3e3ff8c77ebbe480b42c3b53c22fdc27e1456807","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-09-15 03:28:40.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/07/737707/24/check/DellEMC_SC_FC/144f47c/DellEMC_SC_FC/470 : SUCCESS in 1h 07m 49s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/07/737707/24/check/DellEMC_SC_iSCSI/fb1e435/DellEMC_SC_iSCSI/439 : SUCCESS in 1h 28m 33s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":24},{"id":"82dab536fcb47a967b0e12bd11e9d352869178ea","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 03:43:01.000000000","message":"Patch Set 24:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"d7fd559336a01ba87c147647ecd5897402ed3e10","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-09-15 04:30:11.000000000","message":"Patch Set 24:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/737707-24/HPMSA-iSCSI : SUCCESS in 46:35.92\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/737707-24/HPMSA-FC : SUCCESS in 45:35.46","accounts_in_message":[],"_revision_number":24},{"id":"18b5eeef2c4042780b20fb610c280e8d6072373f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-09-15 04:45:04.000000000","message":"Patch Set 24:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-24/Lenovo-iSCSI : SUCCESS in 45:36.18\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-24/Lenovo-FC : FAILURE in 41:00.30","accounts_in_message":[],"_revision_number":24},{"id":"c4a5d24a4bad6d85a2b7135445cd0982399e25c6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-15 05:52:49.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/24/check/dsvm-tempest-zadara-driver/24d0eb1 : FAILURE in 1h 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"78d3603269506007d9cf7074d0f317e6bde6cd0b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-15 06:57:44.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/24/check/DellEMC_XtremIO_FC/2cf9ec1/DellEMC_XtremIO_FC/912 : SUCCESS in 1h 39m 49s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/24/check/DellEMC_XtremIO_iSCSI/b7f8274/DellEMC_XtremIO_iSCSI/875 : SUCCESS in 1h 41m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":24},{"id":"38bc6703b69736a2eaedcf71023351de4ded744d","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-09-15 07:14:00.000000000","message":"Patch Set 24:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-24/Seagate-iSCSI : SUCCESS in 38:50.46\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-24/Seagate-FC : SUCCESS in 37:50.20","accounts_in_message":[],"_revision_number":24},{"id":"d4176bdc6abe01e5e42078465f7b9f154099c24c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-15 07:21:11.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/24/check/DellEMC_Unity_FC/44e2682/DellEMC_Unity_FC/633 : FAILURE in 1h 24m 19s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/24/check/DellEMC_Unity_iSCSI/ce2742c/DellEMC_Unity_iSCSI/606 : FAILURE in 1h 49m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"bb87a8e8459f4f0834a241e50703f9eb523640b7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 07:33:46.000000000","message":"Patch Set 24:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"31b5cd2afb16266c277a89222040ffac265899b9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-15 08:13:53.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"e18d61c2dda006eb5c88d79f1d8891904f2707ca","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-15 08:14:26.000000000","message":"Patch Set 25:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/07/737707/25/check/fujitsu-eternus-dx-iscsi/989d496 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/07/737707/25/check/fujitsu-eternus-dx-fc/2640bd9 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":25},{"id":"3139f1efca0f6ca04bc616378df19c7522c6983c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-09-15 08:15:23.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":25},{"id":"b9d0b6d6226e3b2a25830718bece98e7c08ce2b2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-15 08:19:51.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"831550407c28a4f32dcf93b832b8742d689eb69b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-15 08:35:12.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/07/737707/25/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/edb0370 : FAILURE in 20m 57s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/07/737707/25/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b27c116 : FAILURE in 20m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":25},{"id":"547df1d75b9df2ca2ae67bafbda0772e7113a8d9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 08:47:05.000000000","message":"Patch Set 25:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/25 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"ada070b6cbc77ffe86100594d7bb77d7452b1533","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-15 09:05:02.000000000","message":"Patch Set 25:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/25 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"dcaef1cfeb3869bc61b52e0898d69aa247d83dad","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-15 09:07:23.000000000","message":"Patch Set 25:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/25/iscsi/ : FAILURE in 48m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/25/smb/ : FAILURE in 49m 40s","accounts_in_message":[],"_revision_number":25},{"id":"704fd8d768e4cc462134f036ddc8c64ebb71e249","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-15 09:08:46.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/25/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d8b6b29/d8b6b29.html : FAILURE in 52m 09s","accounts_in_message":[],"_revision_number":25},{"id":"f997148aeb2add882bf1ea2cc06a4a2d711a964d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-15 09:10:00.000000000","message":"Patch Set 25:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/25/check-not-vote/ext-cinder-zuul/ba84562 : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":25},{"id":"e3c813696a651025eda4d2c143088dacc34c2874","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 09:10:07.000000000","message":"Patch Set 25:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/25 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"bb4241733fa6f6cded67ea2769f5bd895d0b7665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:14:56.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/752214055bbe4566bb3a559df9edb127 : FAILURE in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d66c9f71220e4c25ab5cd738b03295f7 : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb611ee05164a5c85c75fc987fa7233 : FAILURE in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/365944e68a0d4f3f8bf9237b637b0c25 : FAILURE in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc4d67f42c544ef84f35435a83c9f0a : FAILURE in 14m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/641d9290d60a49329b9c3e54ffc0a4e2 : POST_FAILURE in 52m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/452aee76ade34831ac90cea6a9da2a8e : FAILURE in 18m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d18c875d5bed4be9bfc82564b9f7795f : SUCCESS in 7m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/44e4823541044bb88a27b584fccd8bb0 : SUCCESS in 12m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc7fc4326a12491d9deadce1fd0a7f57 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a8078cad558c4118a60062ac1d06e299 : FAILURE in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c56888f02b940668e3277a632dd50a6 : SUCCESS in 11m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7d7afa04cc3b4a0a961ee399ced7105a : FAILURE in 4m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8570499c5f4f4610af8eb86614d4ab17 : FAILURE in 15m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/413e743e85cb4d2ca64b6827a8d46ee0 : FAILURE in 7m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/279688554d64464993f65af4fe267b4e : POST_FAILURE in 17m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/315120e0583544db8fcb0ce386148f6e : POST_FAILURE in 25m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c69af3b022cb49d692aba0bbba204c1f : POST_FAILURE in 20m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f2c23db4bc14b60885238f56b704155 : FAILURE in 45m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/418cb2ff49274ee79f364df56f605519 : FAILURE in 23m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d7257a88abb1407ebff8da48de6c5a07 : FAILURE in 19m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d044bbe9de2a43938c2a3fe1b3b39596 : FAILURE in 19m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6660e97df6f48e5be5c47137b3a53a7 : FAILURE in 19m 56s","accounts_in_message":[],"_revision_number":25},{"id":"06c2f91bdbaf599ec20b049ab270385c49e5fad7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-15 09:28:10.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/25/check/cinder-storpool-tempest/050212a/ : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":25},{"id":"404ff35fc0158ceb2d246d5c40db5d8c433e0a6b","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-15 09:48:50.000000000","message":"Patch Set 25:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/b57aa4498552fb76ee2bc027098c1867782f3f26/index.html : SUCCESS","accounts_in_message":[],"_revision_number":25},{"id":"513f848885d5e2b70fe8f3dd1243433734c23ab2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-15 10:21:56.000000000","message":"Patch Set 25:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-25/ns5_nfs : SUCCESS in 2h 7m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-25/ns5_iscsi : SUCCESS in 1h 51m 58s","accounts_in_message":[],"_revision_number":25},{"id":"569ecce1ea91ada8a26e819cf248e98a886b5ebf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-15 10:52:43.000000000","message":"Patch Set 25:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/25/check/tempest-dsvm-full-bionic/898fd00/ : SUCCESS in 1h 38m 04s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/25/check/devstack-plugin-ceph-dsvm-bionic/4640c6f/ : SUCCESS in 1h 58m 59s","accounts_in_message":[],"_revision_number":25},{"id":"9fbfd36750bc6ff9c4423250046bf9d8f6d1f3a2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-15 11:15:24.000000000","message":"Patch Set 25:\n\nStrange failure\n\nERROR: Could not find a version that satisfies the requirement oslo.service\u003d\u003d\u003d2.4.0 (from -c /home/zuul/src/opendev.org/openstack/requirements/upper-constraints.txt (line 36)) (from versions: 0.1.0, 0.2.0, 0.3.0, 0.4.0, 0.5.0, 0.6.0, 0.7.0, 0.8.0, 0.9.0, 0.9.1, 0.10.0, 0.11.0, 0.12.0, 0.13.0, 1.0.0, 1.1.0, 1.2.0, 1.3.0, 1.4.0, 1.5.0, 1.6.0, 1.7.0, 1.8.0, 1.9.0, 1.10.0, 1.11.0, 1.12.0, 1.13.0, 1.14.0, 1.15.0, 1.16.0, 1.16.1, 1.17.0, 1.18.0, 1.19.0, 1.19.1, 1.20.0, 1.21.0, 1.22.0, 1.23.0, 1.24.0, 1.24.1, 1.25.0, 1.25.1, 1.25.2, 1.26.0, 1.27.0, 1.28.0, 1.28.1, 1.29.0, 1.29.1, 1.30.0, 1.31.0, 1.31.1, 1.31.2, 1.31.3, 1.31.4, 1.31.5, 1.31.6, 1.31.7, 1.31.8, 1.32.0, 1.32.1, 1.33.0, 1.34.0, 1.35.0, 1.36.0, 1.37.0, 1.38.0, 1.38.1, 1.39.0, 1.40.0, 1.40.1, 1.40.2, 1.41.0, 1.41.1, 2.0.0, 2.1.0, 2.1.1, 2.1.2, 2.2.0, 2.3.0, 2.3.1, 2.3.2)\n2020-09-15 08:23:28.927865 | ubuntu-bionic | ERROR: No matching distribution found for oslo.service\u003d\u003d\u003d2.4.0 (from -c /home/zuul/src/opendev.org/openstack/requirements/upper-constraints.txt (line 36))\n2020-09-15 08:23:28.927896 | ubuntu-bionic | WARNING: You are using pip version 20.2.2; however, version 20.2.3 is available.","accounts_in_message":[],"_revision_number":25},{"id":"641746c1366987eb009d92a8987cbdd1be6c3256","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 11:28:03.000000000","message":"Patch Set 25:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/25 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"8d1040aad824116d2b7dafce446890560a30646a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-15 11:36:49.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"0c901fd417d8d811a2306c259a6ceea0fffd3875","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-15 11:37:17.000000000","message":"Patch Set 26:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/07/737707/26/check/fujitsu-eternus-dx-iscsi/a617744 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/07/737707/26/check/fujitsu-eternus-dx-fc/34de39c : NOT_REGISTERED","accounts_in_message":[],"_revision_number":26},{"id":"1385885ce9eb1d78aa412e2f67853e9d9b14cc73","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-15 11:37:20.000000000","message":"Patch Set 26:\n\n\u003e (1 comment)\n \u003e \n \u003e From this patch and the ML, it doesn\u0027t look like there\u0027s a lot of\n \u003e interest in the database question.  I think we should keep it\n \u003e simple, then, and hard-delete from the default_types table.\n\nDone","accounts_in_message":[],"_revision_number":26},{"id":"cc7e1e723c3bc5039442eb36c8272c78b805e593","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-09-15 11:40:34.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":26},{"id":"93d3237638ee823ccbd7e3a1f1e2f52fabfdb3f9","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-15 11:58:55.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/07/737707/26/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ae9452f : FAILURE in 21m 42s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/07/737707/26/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/771d7b4 : FAILURE in 20m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":26},{"id":"710d91c2f678c285cb9c819a50646faa0afe3e8a","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-15 12:01:44.000000000","message":"Patch Set 26:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/4a96765271c61bacfb0b0805c4811b28e015cb95/index.html : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"e64ad0dc516f0a9071273647dc57686e97802e4b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-15 12:15:15.000000000","message":"Patch Set 26:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/26/check-not-vote/ext-cinder-zuul/06512d5 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":26},{"id":"52d95e32b1051a8674ba3a5435aaecd0ad8facad","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-15 12:17:21.000000000","message":"Patch Set 26:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"d701b7fdb378d9881a3ef39b1d905dd790f7436d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 12:24:01.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"90828ab30cc418f777ad60165028b8514232e8fa","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-15 12:24:37.000000000","message":"Patch Set 26:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/26/iscsi/ : FAILURE in 47m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/26/smb/ : FAILURE in 27m 38s","accounts_in_message":[],"_revision_number":26},{"id":"e02110ac19dd9bd54afc42deaa3cda2d123c0a80","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-15 12:32:30.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/26/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/881899b/881899b.html : FAILURE in 53m 33s","accounts_in_message":[],"_revision_number":26},{"id":"cd9058adc21eb48d765e9debe99f6decc99b1c26","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 12:40:20.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"14378d2c4db7693b523a6537a957a01a36bb66c4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-15 12:54:25.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/26/check/cinder-storpool-tempest/2268f9c/ : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":26},{"id":"34cc8a33d53bacb717a0fcac6673d9c8d9f66146","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-15 13:18:20.000000000","message":"Patch Set 26:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/26/check/tempest-dsvm-full-bionic/e932a43/ : SUCCESS in 1h 31m 41s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/26/check/devstack-plugin-ceph-dsvm-bionic/de6c263/ : SUCCESS in 1h 32m 37s","accounts_in_message":[],"_revision_number":26},{"id":"2705e20d38c6fa9f08daf25cb3be2bdd2e948257","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-15 13:40:20.000000000","message":"Patch Set 26:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-26/ns5_nfs : SUCCESS in 1h 47m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-26/ns5_iscsi : SUCCESS in 2h 3m 8s","accounts_in_message":[],"_revision_number":26},{"id":"e2d6c6001ffa761e6bd7d1292ac89e85c2bbbba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 13:47:30.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d2a3207973ba4cecb8fa53f471dcb01e : POST_FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd2eeef4ef942ff93d27f6d36fce263 : FAILURE in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20fc05abdb5d4c17a6d4fb456cf8fd0a : FAILURE in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e6810550c234e9f9240e3ffc39f0c89 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7e2404dad5467d8a25c87bd1a090af : POST_FAILURE in 19m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/a418446e57464e6d8293b27dc247deca : SUCCESS in 1h 01m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b01dbc896d7a487bb55cd835bfa53bd5 : SUCCESS in 1h 18m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75ab1f1b78ee4a5e92ee1e643d048f66 : POST_FAILURE in 7m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e14ce74cbeb43609184bd6023e96d8e : SUCCESS in 12m 10s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/21c5cfc099d042d8bbe672b3d69b2b03 : FAILURE in 3m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1686a28ed114441bd7db86389113346 : FAILURE in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dcb0fefd1264e2e9a540fae006c3a40 : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d59522da11964fda80cdf3714f00c5c8 : POST_FAILURE in 3m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d84066a40bd142faa8588eca2165bbc8 : SUCCESS in 39m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d558e5a0a5db4228b8fb4db04d6a6071 : FAILURE in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d666fb56180243a49b78d8f3143b7b7e : SUCCESS in 1h 10m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1f9f3a6758e14d27809f65ad84ea65e2 : POST_FAILURE in 21m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a8865ee530be4f99b1fcac3f0131c03c : POST_FAILURE in 13m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/410cb10ff612470f85f056632198a4ed : SUCCESS in 1h 09m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/57cb560bba48472bb7642cbb0e6b8cff : FAILURE in 12m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/447c1a84932243e48089e68bcdba5ba9 : FAILURE in 13m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da3a34cd823f480aa5c7ad3dc19a8017 : SUCCESS in 2h 04m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/974ba0e4076d412ca80effedbde020ec : FAILURE in 12m 37s","accounts_in_message":[],"_revision_number":26},{"id":"c530f1e2ba305ac7fb4cbb3b03f9f16d5c45f8cd","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-09-15 14:13:20.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/26/check/nimble-iscsi-driver-dsvm-volume/27e07cd : ABORTED in 2h 36m 13s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/26/check/nimble-fc-driver-dsvm-volume/e0f3f31 : SUCCESS in 2h 27m 03s","accounts_in_message":[],"_revision_number":26},{"id":"ff9da7dbe4977f7701c5e8e53052a17e60e78c0f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 14:28:48.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"3c55608ce2e760b4b818443e7772887992d1c1c7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-15 14:50:17.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_Unity_FC/723d30a/DellEMC_Unity_FC/638 : SUCCESS in 1h 45m 04s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_Unity_iSCSI/d824deb/DellEMC_Unity_iSCSI/611 : SUCCESS in 1h 33m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"776bb2c3a6edbea107ecf647d72e5c240cf51207","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 14:58:52.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"9ddc0c9faee4f9c8e63d93608a4cb178db80234b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-15 15:02:57.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_XtremIO_FC/c4d1d62/DellEMC_XtremIO_FC/917 : SUCCESS in 1h 33m 18s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_XtremIO_iSCSI/ba59348/DellEMC_XtremIO_iSCSI/880 : SUCCESS in 1h 43m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":26},{"id":"c80a3cf20ce5349a3fedcf22a1f6cf216c0d2b60","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 15:39:58.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"1a35c7526ef99b2047a691ec90c941828cb5275b","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-09-15 15:42:03.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"019688f573e3e0342d455bc5e2ccaf7522151b1c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2020-09-15 16:01:05.000000000","message":"Patch Set 24:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/07/737707/24/fc : SUCCESS in 0h 48m 00s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/07/737707/24/iscsi : SUCCESS in 0h 58m 57s","accounts_in_message":[],"_revision_number":24},{"id":"c85ebba8170f098f5bb5eb3bcce33d5d91f65937","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-15 16:03:40.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/07/737707/26/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/62e6d49 : FAILURE in 21m 23s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/07/737707/26/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/61c1f6b : FAILURE in 20m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":26},{"id":"ea74c86000c50fa7f503348c0858e8511a60e55c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-15 16:13:29.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/26/check/dsvm-tempest-zadara-driver/47f9721 : FAILURE in 1h 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"621a1620835ec27fedf402f4bb16a730837a014a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 16:18:08.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"b585fb53d2b5ae58a05d79dd4a98a97036eb7f58","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-15 16:31:14.000000000","message":"Patch Set 26:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/26/iscsi/ : FAILURE in 48m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/26/smb/ : FAILURE in 31m 56s","accounts_in_message":[],"_revision_number":26},{"id":"b6f7a1a8a155022c5ef709386fc10f21b8974245","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-15 16:33:06.000000000","message":"Patch Set 26:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"332018412f2e9fe14531ac1676f276023aaa1a71","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-15 16:56:26.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/26/check/cinder-storpool-tempest/d4279de/ : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":26},{"id":"d817c5f7404d3734e29f7e3e98b54ebbc9e48d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 16:59:37.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/20e5ea9e238a4ab2a16cc99a00784e18 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a72a7256b4a467890efcfa3e422bc96 : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5618683e26944b84be12b43e912e1a00 : FAILURE in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d2480e409134c97b72aec830409a410 : FAILURE in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/490db946b5e6479f8dbd63a3e80b1fbb : FAILURE in 13m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/73f948f7189c46e49721645db827f4e6 : FAILURE in 1h 02m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f01d01510254448aa7e6c751e9b5b6a : FAILURE in 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4f4e15b690e4da1b418f07ef4d41b70 : SUCCESS in 11m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ff00ce214eb7465f9b754507b59dde6b : FAILURE in 7m 24s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/66e297297eff4247b5bc09b91b70e2c6 : FAILURE in 5m 13s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ae8d61392abc47db96a51f5b45de6628 : FAILURE in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0df172ea56224b07b414b93203a6c99d : POST_FAILURE in 12m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/304dcbe9cfda4b6db26177ca95180f00 : FAILURE in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b719ac0ba5ba48b8b642ff83f4393d1d : FAILURE in 18m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/171993863cf248bc8a10030916279ce6 : FAILURE in 8m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c9b1aca68783451e8e3d1cbb2e9a7d22 : SUCCESS in 1h 04m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/135b63d1b6084a9a838e2e7275aa5f42 : FAILURE in 1h 09m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/64165b6b99d347cf9c568bf43929ae7f : POST_FAILURE in 19m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2b50eafb99d34674a729aa4ed619ed3a : FAILURE in 1h 04m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8381235f9c848a6a41f5627c49d6ef0 : FAILURE in 13m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9abb9c8e82f347ea93bac397e04a3e2a : FAILURE in 15m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/77209aff7555442a84e1221790ba8e91 : FAILURE in 23m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cd0bb89564341c58f69b5299b10ae68 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":26},{"id":"63c7ea95faae9d8513deea76bcad63a724abe720","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 17:29:13.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"b2a50b784e68c71a893e651a5e0ea3fd0faaff4f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-15 17:41:12.000000000","message":"Patch Set 26:\n\nrecheck maybe we\u0027ll get lucky","accounts_in_message":[],"_revision_number":26},{"id":"05ab2f3bd31f3635e5cc03da6d9cb8096c4969d7","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-15 18:21:52.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_PowerMAX_FC/0505c05/DellEMC_PowerMAX_FC/663 : FAILURE in 3h 47m 22s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_PowerMAX_iSCSI/17b5013/DellEMC_PowerMAX_iSCSI/582 : SUCCESS in 4h 32m 20s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":26},{"id":"558f642d4d312a9d0bc3aac7c23749d911891c3c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-15 18:27:38.000000000","message":"Patch Set 26:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/26/iscsi/ : FAILURE in 45m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/26/smb/ : FAILURE in 46m 00s","accounts_in_message":[],"_revision_number":26},{"id":"17bdb6e6fbb157a54f72e2af4f2899822f04615a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-15 18:34:04.000000000","message":"Patch Set 26:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/26/check/tempest-dsvm-full-bionic/74ca6df/ : SUCCESS in 1h 40m 21s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/26/check/devstack-plugin-ceph-dsvm-bionic/f162a10/ : SUCCESS in 1h 47m 25s","accounts_in_message":[],"_revision_number":26},{"id":"4dee3c414430b74ab1247c54f4bdce308edef3a3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-15 18:56:03.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/26/check/cinder-storpool-tempest/cdf0cd1/ : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":26},{"id":"07950cc72d4ed1fdf8e4d3c2c5d98659a06a848a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 18:56:10.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"b0f1495c25fbaa73501088cb2dd664545ac43f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 20:12:06.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c8a91bd06cc41b1bd01155a1da56b5e : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2f425e0a07423e824d97db01d699c3 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a192c0556943456abad487d625436a05 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/312a816401f44ceb9bfd125e05cf7238 : FAILURE in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39425c8dffa406d99a60a9e42bb70f4 : FAILURE in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/39f436355dfd471eb9ad542ad5b04618 : SUCCESS in 58m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8d932e153bbf4b4d9e1bdb3db733a5af : SUCCESS in 1h 30m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8bcf100266d48338fcf31a70f144f4e : SUCCESS in 7m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e0c79041b2794d4abadf21dc94a8b454 : FAILURE in 3m 21s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b6c2a77249d4ad6843a1e8c55457cd8 : FAILURE in 3m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fc1ff280e5d47db8be2e66c3ecf4750 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dcdd6d969d84221b425b8d52a4d4e06 : SUCCESS in 9m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/96f5b67c95654cecb64bfa37b13b9913 : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bd807c56d705406ea7cb3772c043d3f4 : SUCCESS in 54m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2fdfd006056e4225808c1283dc594c8a : FAILURE in 8m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b3aeafaf2c0647fb92605207b489ed76 : SUCCESS in 56m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/363eaba024f7466381cddabbf4fd5d20 : FAILURE in 1h 41m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3fb52641ca0344f5b066e3de8c1875ce : SUCCESS in 1h 20m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4054873d3ade4433af6d28f29dd4bd57 : FAILURE in 1h 21m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8d86e1ecb182484aa40741f3351698da : FAILURE in 12m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b7c5e40d16740329152785f89af5fee : FAILURE in 15m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/142b4adc90ef48f780b479df432b80f1 : FAILURE in 2h 29m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ec6175894924aa4b526a654c4e16b79 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":26},{"id":"7f9943426f1e19752c606b3d9f834a4d17654732","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-15 21:56:46.000000000","message":"Patch Set 26:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/26/check-not-vote/ext-cinder-zuul/91af726 : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":26},{"id":"b0d9301c1e699a8929fe659517e3479fb52425b5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-15 22:38:29.000000000","message":"Patch Set 26: Code-Review+2\n\n(8 comments)\n\nThis looks good to me.  Thanks for your patience, Rajat.  Some comments inline, but nothing that can\u0027t wait for a followup.","accounts_in_message":[],"_revision_number":26},{"id":"df6bf044dbd4aef1b196860dfb207f05cf3a3f3a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-15 22:38:39.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"7659260576a6d45a1cdf547d3a45cbbaf63477f2","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-15 22:47:37.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/26/check/dsvm-tempest-zadara-driver/39635b0 : FAILURE in 1h 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"07f781811caeb55b7b01a563fcc3285babd6097c","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-15 22:59:49.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/07/737707/26/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b30ae05 : FAILURE in 20m 49s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/07/737707/26/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/48d535f : FAILURE in 20m 29s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":26},{"id":"2fb18b859591b77e2fa484edd934b2e5445ceb6d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-15 23:26:00.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"2c4ecca6dc490b150590b82bb11a9af6dc3996e4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-15 23:27:11.000000000","message":"Patch Set 26:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/26/iscsi/ : FAILURE in 48m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/26/smb/ : FAILURE in 47m 54s","accounts_in_message":[],"_revision_number":26},{"id":"d7d2b2a6d8573f61aacf6705ba08195965cde134","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-15 23:29:25.000000000","message":"Patch Set 26:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"58459a02044cc77b555fe8a4f4e4ce0a8f80ecd4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-15 23:31:56.000000000","message":"Patch Set 26:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/26/check-not-vote/ext-cinder-zuul/53ba63e : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":26},{"id":"b33adeeb19dc1154dd3c3596ebb9e8e34ecb299d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-15 23:46:45.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/26/check/cinder-storpool-tempest/10d3f9d/ : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":26},{"id":"f674841bbb363d7665b2dd90acb80051fa0721cf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-16 00:24:35.000000000","message":"Patch Set 26:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/26/check/tempest-dsvm-full-bionic/8096e43/ : SUCCESS in 1h 24m 29s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/26/check/devstack-plugin-ceph-dsvm-bionic/9263511/ : SUCCESS in 1h 37m 18s","accounts_in_message":[],"_revision_number":26},{"id":"964a703aec193410df3df83345da7dada1c31d89","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 00:25:11.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"deecab449c3022eff28555f4161119010fef5b75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 00:50:25.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2f099249b7c349ad91c09b171f9a1b37 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77654b9e462941abb40ee0d521e9cedb : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe25195658de4c5f9e28bbde38213191 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10eb28c14ef24ef3bc0bf1c1d9198365 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1688eaadceee499485490fbb64cfe679 : SUCCESS in 20m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/9b33a0fb0fb84200862cc064b1830ea1 : SUCCESS in 59m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70dd820edff341a69d01e96fe7a857ab : SUCCESS in 1h 07m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d00a4dc7416d4d02ad9dbbceac16bffc : SUCCESS in 7m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d8613df232254759a718895fcdc221d8 : SUCCESS in 11m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bafd2be716c74f289c0a6844af53e705 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/02853fe733f94119aa522111c6bd3913 : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f61ce5158fc49b7acff38e3b3aca7cd : SUCCESS in 10m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e239f5b81f50422b9f9523ff8a0430a1 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f15b77bd55e7468ab1b8080680f3e2cb : SUCCESS in 38m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dbe6105f99ec4d72a24ce6615eea5beb : FAILURE in 4m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e8f7dd94ea6c4c1b901de9d6952e0dc3 : SUCCESS in 1h 00m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0833b1a9e9764b32acb38b10f6d31e5d : FAILURE in 1h 05m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/59766591bbf04fdf978c9e8773fbccba : SUCCESS in 2h 02m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f7e2119f7ba04011b12ea69cfcfa62b4 : SUCCESS in 1h 03m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7810bf6b38234f98aa49d8ebeae4df6c : SUCCESS in 1h 07m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5624110d45f14f968537d270bfd6c237 : SUCCESS in 1h 12m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a690f84482e4226b39e13792a7a08dc : SUCCESS in 1h 29m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a36b3d6f98d646ec85f833bfa91369c6 : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":26},{"id":"39e91f046975f3a8d22741d206b96608534873f0","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-16 00:59:24.000000000","message":"Patch Set 26:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/26/check/synology-dsm-cinder-iscsi/165326b : FAILURE in 8m 24s","accounts_in_message":[],"_revision_number":26},{"id":"3137dd901efec200c115fff072183952c8d33ee0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-16 01:10:17.000000000","message":"Patch Set 26:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-26 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":26},{"id":"a2f85126a4ef1dfab2b7015d467e86d8051c77c3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 01:34:04.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"050dd9844c9bb610d908d56553d4df8949d46802","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-16 01:55:29.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/26/upstream-solidfire-check/cinder-solidfire-iSCSI/4ef0e02/ : SUCCESS in 1h 03m 48s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":26},{"id":"88ef670251846f19b6598dec7fdb29bfc382dc0b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-09-16 01:57:20.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_PowerStore_iSCSI/8fda7a3/DellEMC_PowerStore_iSCSI/206 : SUCCESS in 1h 01m 57s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_PowerStore_FC/54d02a0/DellEMC_PowerStore_FC/207 : SUCCESS in 1h 06m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":26},{"id":"2d42b89665fe23736b08bc2f519b8b952392628d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 02:00:10.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"3cd010f92a69ba7e47bb363f558dcda4b070ad67","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-09-16 02:18:56.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_SC_FC/5852862/DellEMC_SC_FC/473 : FAILURE in 1h 14m 29s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_SC_iSCSI/43e8da0/DellEMC_SC_iSCSI/442 : SUCCESS in 1h 28m 14s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":26},{"id":"31b95a02c416d919097a8f330e4b627a1613b787","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 02:21:45.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"9780a48b5a9bae73e59c1cc2209b3dab60f825aa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-16 02:38:25.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/26/upstream-check/cinder-cDOT-iSCSI/631ea52/ : SUCCESS in 35m 21s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/26/upstream-check/cinder-cDOT-iSCSI-IPv6/315d1f9/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/26/upstream-check/cinder-cDOT-NFS/7b3d1ad/ : SUCCESS in 1h 42m 08s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/26/upstream-check/cinder-cDOT-NFS-IPv6/c07b0d4/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/26/upstream-check/cinder-cDOT-FCP/4f609ad/ : SUCCESS in 53m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":26},{"id":"b9c2405abff280fc34b60fcd75a0114b3a0cdbae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-16 02:45:56.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_Unity_FC/b08f665/DellEMC_Unity_FC/642 : SUCCESS in 1h 28m 16s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_Unity_iSCSI/f32650c/DellEMC_Unity_iSCSI/615 : SUCCESS in 1h 41m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"f5df49e10694765aff88401ea42ef0ec3de68f72","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 02:46:11.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_XtremIO_FC/d4bb7c5/DellEMC_XtremIO_FC/921 : SUCCESS in 1h 34m 11s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_XtremIO_iSCSI/b8ade6a/DellEMC_XtremIO_iSCSI/884 : SUCCESS in 1h 45m 37s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":26},{"id":"ca704ec55227735cd13d68ddc83ec53e522c779f","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-09-16 02:57:38.000000000","message":"Patch Set 26:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-26/Seagate-iSCSI : SUCCESS in 40:20.71\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-26/Seagate-FC : SUCCESS in 37:54.21","accounts_in_message":[],"_revision_number":26},{"id":"22d48edd0892a82947a16ad6b5ca3aa04c258239","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-09-16 03:19:13.000000000","message":"Patch Set 26:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-26/Lenovo-iSCSI : SUCCESS in 46:11.76\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-26/Lenovo-FC : SUCCESS in 45:26.14","accounts_in_message":[],"_revision_number":26},{"id":"4b3a7725de26ad7f8226124cecc905ccaf538423","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 03:21:03.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"d7fc67294b8d27c8d8fff83c05dd98ab0ee28d05","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-09-16 03:30:26.000000000","message":"Patch Set 26:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/737707-26/HPMSA-iSCSI : SUCCESS in 53:52.41\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/737707-26/HPMSA-FC : SUCCESS in 48:43.41","accounts_in_message":[],"_revision_number":26},{"id":"1421ebe9d942a1344a3198b183ad1d13d6e00d56","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 04:58:04.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_VNX_iSCSI/e651512/DellEMC_VNX_iSCSI/404 : SUCCESS in 2h 48m 25s\n- DellEMC_VNX_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_VNX_FC/a21209a/DellEMC_VNX_FC/418 : SUCCESS in 1h 03m 58s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":26},{"id":"af47c2d121dfe58fc198d91fd52039e139cc4963","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 05:33:51.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"ca6b6d2da0e9a80877332dad33a6b73d5b465026","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-16 08:50:15.000000000","message":"Patch Set 26: Code-Review+2\n\n(5 comments)\n\nLGTM, though we could remove the \"id\" field from the DB","accounts_in_message":[],"_revision_number":26},{"id":"4ab3642ddf41f6e93a90b3d8d82931a2267011c3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 10:45:54.000000000","message":"Patch Set 26:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/26/20200916170254/NEC-FC : SUCCESS in 2h 29m 28s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/737707/26/20200916170254/NEC-ISCSI : FAILURE in 2h 42m 59s","accounts_in_message":[],"_revision_number":26},{"id":"0d6d6508140d02c455cb89e4b62720dadd621032","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-16 12:15:34.000000000","message":"Patch Set 26:\n\n(1 comment)\n\nComment about the model inline.  I agree with Gorka that it would be a good idea to make the project_id the primary key (it will get us a free index on that column, too) and removing the \u0027id\u0027.  If you make that revision, you can still hold the other stuff I mentioned in my previous contents for a followup.","accounts_in_message":[],"_revision_number":26},{"id":"5f5ef7b1211b4e5f4b2e8506231b621aaa3e104b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 12:28:04.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/07/737707/26/check/DellEMC_PowerMAX_FC/2cba519/DellEMC_PowerMAX_FC/667 : SUCCESS in 3h 52m 17s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/07/737707/26/check/DellEMC_PowerMAX_iSCSI/7a54622/DellEMC_PowerMAX_iSCSI/586 : SUCCESS in 4h 23m 44s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":26},{"id":"dfb32a6863446a82571df1df4f6e74a76d1c41d6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 13:12:38.000000000","message":"Patch Set 26:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"a2fd55d456fc6a52a7eb7f07953b797266dbbff5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-16 13:38:58.000000000","message":"Patch Set 26: Workflow-1\n\nHolding for now, Rajat is in the middle of some final revisions.","accounts_in_message":[],"_revision_number":26},{"id":"640856e5492499d87b625a6249ab5385fa85fa6a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-16 14:06:18.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"e0a5ea99bb313642d385dcf164a09931be087e0a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-16 14:06:42.000000000","message":"Patch Set 27:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/07/737707/27/check/fujitsu-eternus-dx-iscsi/73f88c2 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/07/737707/27/check/fujitsu-eternus-dx-fc/afe6c1e : NOT_REGISTERED","accounts_in_message":[],"_revision_number":27},{"id":"4eeede23b2c8d759b4161c77982fc0ef63867f2a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-16 14:10:20.000000000","message":"Patch Set 27:\n\n(10 comments)","accounts_in_message":[],"_revision_number":27},{"id":"0d300c0a47e8db87024d4a419c919e2657b4a02f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-16 14:29:24.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/07/737707/27/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/374d31d : FAILURE in 21m 31s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/07/737707/27/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/28d09b7 : FAILURE in 21m 02s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":27},{"id":"b955e5e97000df59680d21ca7a27b6bf6571d02b","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 14:42:32.000000000","message":"Patch Set 27:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/e63cb8548a64e67bad39409c15aa210daea01d7b/index.html : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"6396d94c4dc7da861772def7c2c3d05eaa24eeb3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-16 14:54:10.000000000","message":"Patch Set 27:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/737707/27/iscsi/ : FAILURE in 47m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/737707/27/smb/ : FAILURE in 46m 47s","accounts_in_message":[],"_revision_number":27},{"id":"d64be3a5c8da79c5177d4bee95145288fbf4a2ad","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-16 14:57:04.000000000","message":"Patch Set 27:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/737707/27 : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"dfb47105223be167bed7111bf1c79864e647ece9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-16 15:01:49.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/07/737707/27/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/69653f3/69653f3.html : FAILURE in 52m 35s","accounts_in_message":[],"_revision_number":27},{"id":"d042f747b28e54bb7c4d71811b3a46feb73f31f5","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-09-16 15:38:43.000000000","message":"Patch Set 27: Code-Review+2\n\nI think this is a good change.  No major concerns.","accounts_in_message":[],"_revision_number":27},{"id":"93290ffddc33885c8c7c5d76dafee869280d6976","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-09-16 15:56:58.000000000","message":"Patch Set 27: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":27},{"id":"04c2b98272593a8eb9fa962534ca6955656c568d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-16 15:57:12.000000000","message":"Patch Set 27:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-737707-27/ns5_nfs : SUCCESS in 1h 50m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-737707-27/ns5_iscsi : SUCCESS in 1h 50m 28s","accounts_in_message":[],"_revision_number":27},{"id":"f4eed998e3515fe2e86f5229963508d1a39d1460","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-16 16:44:43.000000000","message":"Patch Set 27:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/27/check/tempest-dsvm-full-bionic/eb32350/ : SUCCESS in 1h 41m 48s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/737707/27/check/devstack-plugin-ceph-dsvm-bionic/6371b2b/ : SUCCESS in 1h 59m 43s","accounts_in_message":[],"_revision_number":27},{"id":"3403b9470d0571fcfc99d1a0ac785ff60d63fd55","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 17:28:09.000000000","message":"Patch Set 27:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/27 : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"d87132c0228f7c03c626279247458b56a3499af0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 17:39:16.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b9a9a429978d4306b640ee01125977e0 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf099109978a45ddb97fbb2e0dd3018c : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6eb13d3552a45ec8c9645d9569f3221 : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56eddb71e47444babfe3caf27cb75d67 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098499622944475b81688142b0dee153 : SUCCESS in 21m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/fdb89e31224c4b92bebf56708e036dcb : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f334becfaf7e4e319d4d4593dbc03eb6 : SUCCESS in 1h 18m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7d5288dda444db5b8e8be32793c27ee : SUCCESS in 11m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/40da4737a96a4af89d1fde3c894ca13e : SUCCESS in 16m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce7bda230fee40a9b78c0d0a89749f30 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc64ab929e8a42499697f39a040eea4c : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53f536a210c5480b835e8e07b3acf2d5 : SUCCESS in 12m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c12fcc6e52bf4521972fa55a3a29bdcd : SUCCESS in 8m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/54c2a12dd19e4e5eab9485361a6954fb : SUCCESS in 46m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a492e66a862484da6e5b13af3c1e443 : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0df0c9deb9b4156a3daad823e62e147 : SUCCESS in 1h 01m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04bc40cf8ce44db7827333d5fc730ec0 : FAILURE in 1h 20m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/72b4d64adf9542148f5ceba00d02a0f5 : SUCCESS in 1h 30m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a6918930863343488ef1968d7fabc9d7 : SUCCESS in 1h 14m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bf64dc43721747b38113b6eaaf279056 : SUCCESS in 1h 19m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e8590626f00b49cf9402a8f698001dea : SUCCESS in 1h 14m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc9dd75c136140609d3d9df831911b2d : SUCCESS in 2h 06m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d04d86147d164f2b939eb5d66e44a571 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":27},{"id":"eaa076c3d2366bdc84650fd3ebe5500630b923d8","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-16 17:48:09.000000000","message":"Patch Set 27:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/07/737707/27/check/synology-dsm-cinder-iscsi/c84cea9 : FAILURE in 8m 15s","accounts_in_message":[],"_revision_number":27},{"id":"1410e1a6291c499e01989bed3acaf7d5f47018ea","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 17:54:54.000000000","message":"Patch Set 27:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/27 : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"7eb4da250ba91eb19893ce7627da11a7b9337dd5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-16 17:59:29.000000000","message":"Patch Set 27:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-07-737707-27 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":27},{"id":"3b7ed426c5a9466490d51c24305a670e652d8daf","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-09-16 17:59:51.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/27/check/nimble-iscsi-driver-dsvm-volume/30bd71c : FAILURE in 3h 53m 15s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/07/737707/27/check/nimble-fc-driver-dsvm-volume/41692f2 : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":27},{"id":"8cde47b68f0c3c228f00196c6c0ff1c3e38a1dfc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2020-09-16 18:01:07.000000000","message":"Patch Set 26:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/07/737707/26/fc : SUCCESS in 0h 43m 08s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/07/737707/26/iscsi : SUCCESS in 0h 59m 35s","accounts_in_message":[],"_revision_number":26},{"id":"18f2ab089e9b7618b418bc991be8cd68e6d9c787","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-16 18:15:07.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1\n\nLGTM!","accounts_in_message":[],"_revision_number":27},{"id":"3dd269ae7a23b7b287728d482da059844566ddb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 18:15:36.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"790963d3764a705637c00402420c88ff3a32b460","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-16 18:36:04.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/737707/27/upstream-check/cinder-cDOT-iSCSI/cd49362/ : SUCCESS in 38m 00s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/737707/27/upstream-check/cinder-cDOT-iSCSI-IPv6/9881bdd/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/737707/27/upstream-check/cinder-cDOT-NFS/080ec3a/ : SUCCESS in 38m 56s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/737707/27/upstream-check/cinder-cDOT-NFS-IPv6/3a1e1fd/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/737707/27/upstream-check/cinder-cDOT-FCP/78c303d/ : SUCCESS in 55m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":27},{"id":"393ec92e4ef927d522e95ccbec92b50c01da0481","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-16 18:50:20.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/737707/27/upstream-solidfire-check/cinder-solidfire-iSCSI/b4cbd14/ : FAILURE in 1h 09m 30s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":27},{"id":"5f3a2221ecb0ecea3fffa77d5dad17e598ceeefe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-16 19:05:44.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/07/737707/27/check/cinder-storpool-tempest/59e3050/ : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":27},{"id":"b86bbe6d6733d67cdf36b6725314b2a3aa20725d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-16 20:20:22.000000000","message":"Patch Set 27:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/07/737707/27 : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"372c33d56709c98b263a6b394b9102b6f05018bd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-16 20:25:33.000000000","message":"Patch Set 27:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/07/737707/27/check-not-vote/ext-cinder-zuul/3e3495e : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":27},{"id":"230ab2dc13e76f51c9f55e317a54e6e85b720e3f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-16 20:30:01.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/737707/27/check/dsvm-tempest-zadara-driver/4f6e572 : FAILURE in 1h 15m 57s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d7c06bff20e6d095c1dd197c5ee8a539a25ab2c6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 21:18:19.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/07/737707/27/check/DellEMC_VNX_iSCSI/8661065/DellEMC_VNX_iSCSI/415 : FAILURE in 27m 17s\n- DellEMC_VNX_FC http://publiclogs.emc.com/07/737707/27/check/DellEMC_VNX_FC/f95924b/DellEMC_VNX_FC/429 : SUCCESS in 1h 01m 55s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":27},{"id":"8cb0543eaa7b11b5cd4bd71f75dac257228e1d01","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-09-16 21:29:39.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/07/737707/27/check/DellEMC_PowerStore_iSCSI/755b4c1/DellEMC_PowerStore_iSCSI/217 : SUCCESS in 1h 00m 49s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/07/737707/27/check/DellEMC_PowerStore_FC/6df0602/DellEMC_PowerStore_FC/218 : SUCCESS in 1h 02m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":27},{"id":"b5d4706c7472782bb1f53a18bb932aee991e4113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:49:13.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/af14cb125e0d4530a43159b4f4156ac4 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6bb8229ef3f4dcea48cd26020cf902a : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ada043016f64c508a0c89b6c1c480fb : SUCCESS in 11m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77d360510be1466fa949bbead8f7c279 : SUCCESS in 10m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8caaf89fe349e9a70702ae2989b299 : SUCCESS in 25m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/2a0dcf9a59cb4b4a8ac94690a7162dda : SUCCESS in 57m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9f8c86efecd4b60b4e0f51accb26228 : SUCCESS in 59m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6ba739031784e7d801567f837c234fa : SUCCESS in 7m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a1170d7d3e849ea8f6f5d03bf4e1c13 : SUCCESS in 9m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/881e9cec87114ba0ab97befd6f24f735 : SUCCESS in 1h 01m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7e2d7a76599c4fff8a09ad6adc467c80 : SUCCESS in 1h 02m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16e878e13f7b4f24a5ea442d738edc2b : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":27},{"id":"67df818317860aa36b6708a301be4d55e827ffb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:49:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":27},{"id":"9e6a7051b122f83d7c48f36bcac8c78d8f8a6f5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:00:00.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6bff8b4c81d14fb699d26d373db96abe : SUCCESS in 1m 30s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af54143bedb6421b9ce7998f2fc0b4ec : SUCCESS in 2m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9263e12c7a734ff58651af8859456645 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":27},{"id":"7143c2c77a859f74963b46a8fba95a1d0c6912ef","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-09-17 02:04:49.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/07/737707/27/check/DellEMC_SC_FC/2231591/DellEMC_SC_FC/484 : SUCCESS in 1h 17m 07s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/07/737707/27/check/DellEMC_SC_iSCSI/777edbf/DellEMC_SC_iSCSI/453 : SUCCESS in 1h 36m 58s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":27},{"id":"5ed8271fd02f5cc1dd92411df526d146a6561287","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-17 03:41:12.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/07/737707/27/check/DellEMC_Unity_FC/f2cd8a3/DellEMC_Unity_FC/656 : SUCCESS in 1h 28m 31s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/07/737707/27/check/DellEMC_Unity_iSCSI/624c05a/DellEMC_Unity_iSCSI/626 : FAILURE in 1h 42m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":27},{"id":"de9ad06e6bfa816d595b16f5a0c7a0629ffeaed9","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-09-17 04:48:16.000000000","message":"Patch Set 27:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/737707-27/Seagate-iSCSI : SUCCESS in 40:13.40\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/737707-27/Seagate-FC : SUCCESS in 39:04.47","accounts_in_message":[],"_revision_number":27},{"id":"464aab93e94bb138fb55fa3c77faee2b294e665f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-09-17 08:13:40.000000000","message":"Patch Set 27:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/737707-27/HPMSA-iSCSI : SUCCESS in 44:28.69\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/737707-27/HPMSA-FC : SUCCESS in 42:46.06","accounts_in_message":[],"_revision_number":27},{"id":"ceb6f1d10c4a2c7f7d5f6d8d13893450325b0218","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-09-17 08:15:06.000000000","message":"Patch Set 27:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/737707-27/Lenovo-iSCSI : SUCCESS in 48:13.16\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/737707-27/Lenovo-FC : SUCCESS in 44:42.91","accounts_in_message":[],"_revision_number":27},{"id":"6c3018fb64ef26d854b6488178032c5bd9b202f8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-09-17 09:09:50.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/07/737707/27/check/DellEMC_PowerFlex/380684c/DellEMC_PowerFlex/495 : SUCCESS in 2h 10m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"e63cb8548a64e67bad39409c15aa210daea01d7b","revisions":{"dd08d7c2857907d2d55f62a0b98915084fb88756":{"kind":"REWORK","_number":1,"created":"2020-06-24 08:19:44.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/1"}}},"commit":{"parents":[{"commit":"d1ef0edd05285ca4ad1e5e31b7a2cf5063b2ec4c","subject":"Merge \"Add revert to snapshot support for Pure Storage drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1ef0edd05285ca4ad1e5e31b7a2cf5063b2ec4c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"subject":"WIP: default type overrides","message":"WIP: default type overrides\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd08d7c2857907d2d55f62a0b98915084fb88756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd08d7c2857907d2d55f62a0b98915084fb88756"}]},"branch":"refs/heads/master"},"646f24e1419d340d286bf8441d39b2f4f893a69c":{"kind":"REWORK","_number":2,"created":"2020-06-26 15:02:20.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/2"}}},"commit":{"parents":[{"commit":"d1ef0edd05285ca4ad1e5e31b7a2cf5063b2ec4c","subject":"Merge \"Add revert to snapshot support for Pure Storage drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1ef0edd05285ca4ad1e5e31b7a2cf5063b2ec4c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-26 15:02:00.000000000","tz":0},"subject":"WIP: default type overrides","message":"WIP: default type overrides\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/646f24e1419d340d286bf8441d39b2f4f893a69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/646f24e1419d340d286bf8441d39b2f4f893a69c"}]},"branch":"refs/heads/master"},"7466b565af09c34b3d3afa528774706875c00928":{"kind":"REWORK","_number":3,"created":"2020-07-03 12:09:02.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/3"}}},"commit":{"parents":[{"commit":"d1ef0edd05285ca4ad1e5e31b7a2cf5063b2ec4c","subject":"Merge \"Add revert to snapshot support for Pure Storage drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1ef0edd05285ca4ad1e5e31b7a2cf5063b2ec4c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-07-03 12:07:42.000000000","tz":0},"subject":"WIP: default type overrides","message":"WIP: default type overrides\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7466b565af09c34b3d3afa528774706875c00928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7466b565af09c34b3d3afa528774706875c00928"}]},"branch":"refs/heads/master"},"09f2139557a2ed29c15c846f7191fc49bb6eadf9":{"kind":"REWORK","_number":4,"created":"2020-07-09 06:55:29.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/4"}}},"commit":{"parents":[{"commit":"d1ef0edd05285ca4ad1e5e31b7a2cf5063b2ec4c","subject":"Merge \"Add revert to snapshot support for Pure Storage drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1ef0edd05285ca4ad1e5e31b7a2cf5063b2ec4c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-07-09 06:55:03.000000000","tz":0},"subject":"WIP: default type overrides","message":"WIP: default type overrides\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09f2139557a2ed29c15c846f7191fc49bb6eadf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09f2139557a2ed29c15c846f7191fc49bb6eadf9"}]},"branch":"refs/heads/master"},"267192337a46ffdbf29eccf136c90c3dbca6a499":{"kind":"REWORK","_number":5,"created":"2020-07-14 15:30:18.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/5"}}},"commit":{"parents":[{"commit":"70beaeb78857264b371ad9d0c49be24dc078edfc","subject":"Merge \"Add non-voting code coverage job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70beaeb78857264b371ad9d0c49be24dc078edfc"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-07-14 15:29:27.000000000","tz":0},"subject":"WIP: default type overrides","message":"WIP: default type overrides\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/267192337a46ffdbf29eccf136c90c3dbca6a499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/267192337a46ffdbf29eccf136c90c3dbca6a499"}]},"branch":"refs/heads/master"},"4c69444336f15ab92a342a9b2e853211203eea96":{"kind":"REWORK","_number":6,"created":"2020-07-21 12:42:26.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/6"}}},"commit":{"parents":[{"commit":"70beaeb78857264b371ad9d0c49be24dc078edfc","subject":"Merge \"Add non-voting code coverage job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70beaeb78857264b371ad9d0c49be24dc078edfc"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-07-21 12:42:04.000000000","tz":0},"subject":"WIP: default type overrides","message":"WIP: default type overrides\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c69444336f15ab92a342a9b2e853211203eea96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c69444336f15ab92a342a9b2e853211203eea96"}]},"branch":"refs/heads/master"},"acfaa9ede3386584520e80665b985ee055c7cbd9":{"kind":"REWORK","_number":7,"created":"2020-08-12 15:38:15.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/7"}}},"commit":{"parents":[{"commit":"70beaeb78857264b371ad9d0c49be24dc078edfc","subject":"Merge \"Add non-voting code coverage job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70beaeb78857264b371ad9d0c49be24dc078edfc"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-12 15:31:45.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/acfaa9ede3386584520e80665b985ee055c7cbd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/acfaa9ede3386584520e80665b985ee055c7cbd9"}]},"branch":"refs/heads/master"},"1295878d746698011e295a9f7ff30ae880bf98bf":{"kind":"REWORK","_number":8,"created":"2020-08-13 11:39:55.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/8"}}},"commit":{"parents":[{"commit":"70beaeb78857264b371ad9d0c49be24dc078edfc","subject":"Merge \"Add non-voting code coverage job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70beaeb78857264b371ad9d0c49be24dc078edfc"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-13 11:37:53.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1295878d746698011e295a9f7ff30ae880bf98bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1295878d746698011e295a9f7ff30ae880bf98bf"}]},"branch":"refs/heads/master"},"8a564f1b3e125276e8c2768e878e88ed23215b11":{"kind":"REWORK","_number":9,"created":"2020-08-24 06:01:25.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/9"}}},"commit":{"parents":[{"commit":"c48fdb9f9e8af1e2d99841c41a65e39bdbd509cc","subject":"Merge \"Modify default/delete volume type logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c48fdb9f9e8af1e2d99841c41a65e39bdbd509cc"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-24 06:00:26.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a564f1b3e125276e8c2768e878e88ed23215b11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a564f1b3e125276e8c2768e878e88ed23215b11"}]},"branch":"refs/heads/master"},"149f08f352602678af3a4f9b4144a9b86c2ccbf3":{"kind":"REWORK","_number":10,"created":"2020-08-24 07:03:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/10"}}},"commit":{"parents":[{"commit":"c48fdb9f9e8af1e2d99841c41a65e39bdbd509cc","subject":"Merge \"Modify default/delete volume type logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c48fdb9f9e8af1e2d99841c41a65e39bdbd509cc"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-24 07:03:10.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/149f08f352602678af3a4f9b4144a9b86c2ccbf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/149f08f352602678af3a4f9b4144a9b86c2ccbf3"}]},"branch":"refs/heads/master"},"47f911b742d87bf61ba7720cb90cd47582c47de0":{"kind":"REWORK","_number":11,"created":"2020-08-24 12:13:15.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/11"}}},"commit":{"parents":[{"commit":"c48fdb9f9e8af1e2d99841c41a65e39bdbd509cc","subject":"Merge \"Modify default/delete volume type logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c48fdb9f9e8af1e2d99841c41a65e39bdbd509cc"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-24 12:12:36.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47f911b742d87bf61ba7720cb90cd47582c47de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47f911b742d87bf61ba7720cb90cd47582c47de0"}]},"branch":"refs/heads/master"},"3d0e37fd963be00fb5a3348f1de16c88161ae3e8":{"kind":"REWORK","_number":12,"created":"2020-08-28 14:30:22.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/12"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-28 14:29:46.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d0e37fd963be00fb5a3348f1de16c88161ae3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d0e37fd963be00fb5a3348f1de16c88161ae3e8"}]},"branch":"refs/heads/master"},"cf34f6cb39b5165bdcd8f88af5e9586d7bb6a7ff":{"kind":"REWORK","_number":13,"created":"2020-08-28 20:24:20.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/13"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-28 20:23:42.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf34f6cb39b5165bdcd8f88af5e9586d7bb6a7ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf34f6cb39b5165bdcd8f88af5e9586d7bb6a7ff"}]},"branch":"refs/heads/master"},"82214b5d3317b0913139e184568f15c4649db971":{"kind":"REWORK","_number":14,"created":"2020-08-31 09:54:26.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/14"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-31 09:52:58.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/82214b5d3317b0913139e184568f15c4649db971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/82214b5d3317b0913139e184568f15c4649db971"}]},"branch":"refs/heads/master"},"38564244802684645dd35bf29e60a67713b59a3b":{"kind":"REWORK","_number":15,"created":"2020-09-04 12:01:58.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/15"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-04 12:01:15.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38564244802684645dd35bf29e60a67713b59a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38564244802684645dd35bf29e60a67713b59a3b"}]},"branch":"refs/heads/master"},"dbd8ef20bff571c71c3f64e4c2e3f2ae83b22556":{"kind":"REWORK","_number":16,"created":"2020-09-04 15:25:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/16"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-04 15:25:20.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbd8ef20bff571c71c3f64e4c2e3f2ae83b22556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbd8ef20bff571c71c3f64e4c2e3f2ae83b22556"}]},"branch":"refs/heads/master"},"e044393a536656e9939ee5a512ec4a42de8321f3":{"kind":"REWORK","_number":17,"created":"2020-09-07 14:16:21.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/17"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-07 14:15:33.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e044393a536656e9939ee5a512ec4a42de8321f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e044393a536656e9939ee5a512ec4a42de8321f3"}]},"branch":"refs/heads/master"},"c5a547c560eeb329e9c20d14f438e5e57ffa7ff7":{"kind":"REWORK","_number":18,"created":"2020-09-08 10:38:42.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/18"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-08 10:37:45.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5a547c560eeb329e9c20d14f438e5e57ffa7ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5a547c560eeb329e9c20d14f438e5e57ffa7ff7"}]},"branch":"refs/heads/master"},"c0a1b6efce83b8532220b7b468f3176afba7bcaa":{"kind":"REWORK","_number":19,"created":"2020-09-08 11:51:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/19"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-08 11:51:11.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0a1b6efce83b8532220b7b468f3176afba7bcaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0a1b6efce83b8532220b7b468f3176afba7bcaa"}]},"branch":"refs/heads/master"},"d786963e93865051a503eed26ed597021f9dd95f":{"kind":"REWORK","_number":20,"created":"2020-09-08 14:40:08.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/20"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-08 14:39:20.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d786963e93865051a503eed26ed597021f9dd95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d786963e93865051a503eed26ed597021f9dd95f"}]},"branch":"refs/heads/master"},"c84b9113a38ccf7a95cb9565420b9f87ec1fa2e5":{"kind":"REWORK","_number":21,"created":"2020-09-08 16:34:12.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/21"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-08 16:33:38.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c84b9113a38ccf7a95cb9565420b9f87ec1fa2e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c84b9113a38ccf7a95cb9565420b9f87ec1fa2e5"}]},"branch":"refs/heads/master"},"1d37f45ceda0e426366ce893a6ee01ffa1b967f6":{"kind":"REWORK","_number":22,"created":"2020-09-09 15:36:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/22"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-09 15:36:10.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d37f45ceda0e426366ce893a6ee01ffa1b967f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d37f45ceda0e426366ce893a6ee01ffa1b967f6"}]},"branch":"refs/heads/master"},"5a370f2cb9a39a9636b70f6b4a21e72cc331abb0":{"kind":"REWORK","_number":23,"created":"2020-09-10 11:36:49.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/23"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-10 11:36:02.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a370f2cb9a39a9636b70f6b4a21e72cc331abb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a370f2cb9a39a9636b70f6b4a21e72cc331abb0"}]},"branch":"refs/heads/master"},"06294f4906fa721bac23d86aa29aafa7c58fdcd4":{"kind":"REWORK","_number":24,"created":"2020-09-14 20:08:33.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/24"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-14 20:07:52.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) Policies that defaults to system admin for all operations\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system admin by default but a new rule is available\nto allow this for domain and project admins as well.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06294f4906fa721bac23d86aa29aafa7c58fdcd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06294f4906fa721bac23d86aa29aafa7c58fdcd4"}]},"branch":"refs/heads/master"},"b57aa4498552fb76ee2bc027098c1867782f3f26":{"kind":"REWORK","_number":25,"created":"2020-09-15 08:13:53.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/25"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-15 08:12:51.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) All policies (except get_all) default to system/domain/project admin\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system/domain/project admin by default except get_all\npolicy which defaults to system admin.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b57aa4498552fb76ee2bc027098c1867782f3f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b57aa4498552fb76ee2bc027098c1867782f3f26"}]},"branch":"refs/heads/master"},"4a96765271c61bacfb0b0805c4811b28e015cb95":{"kind":"REWORK","_number":26,"created":"2020-09-15 11:36:49.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/26"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-15 11:35:58.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) All policies (except get_all) default to system/domain/project admin\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system/domain/project admin by default except get_all\npolicy which defaults to system admin.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a96765271c61bacfb0b0805c4811b28e015cb95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a96765271c61bacfb0b0805c4811b28e015cb95"}]},"branch":"refs/heads/master"},"e63cb8548a64e67bad39409c15aa210daea01d7b":{"kind":"REWORK","_number":27,"created":"2020-09-16 14:06:18.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/07/737707/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/737707/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/737707/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/737707/27"}}},"commit":{"parents":[{"commit":"63a8f2e1bf00829000ac45e4f20b2a8022b48928","subject":"Merge \"Show cluster_name in volume details\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a8f2e1bf00829000ac45e4f20b2a8022b48928"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-06-24 08:18:51.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-09-16 14:05:31.000000000","tz":0},"subject":"Default type overrides","message":"Default type overrides\n\nThis patch adds a feature by which we allow setting default volume types\nfor projects.\nThe following changes are made to achieve the feature:\n\n1) Add 4 set of APIs, set, get, get_all, unset volume type\n2) All policies (except get_all) default to system/domain/project admin\n3) Preference order: project default, conf default\n4) Logic to not allow deletion of default type\n\nWe validate set, get and unset APIs with keystone to verify a valid\nproject id is passed in the request and user has proper authorization\nrights to show the project.\n\nThe policies are system/domain/project admin by default except get_all\npolicy which defaults to system admin.\n\nImplements: Blueprint multiple-default-volume-types\n\nChange-Id: Idcc949ed6adbaea0c2337fac83014998b81ff1f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e63cb8548a64e67bad39409c15aa210daea01d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e63cb8548a64e67bad39409c15aa210daea01d7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
