)]}'
{"id":"openstack%2Fcinder~738687","triplet_id":"openstack%2Fcinder~master~Icbc1fbb1c9dcf3312860e504adb134811259c817","project":"openstack/cinder","branch":"master","topic":"add-coverage-job","hashtags":[],"change_id":"Icbc1fbb1c9dcf3312860e504adb134811259c817","subject":"Add non-voting code coverage job","status":"MERGED","created":"2020-06-30 20:02:33.000000000","updated":"2020-07-09 05:11:06.000000000","submitted":"2020-07-08 23:38:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738687-1594251515535-4780df3a","meta_rev_id":"6d1e785bdf0463c46a9142fa04b8c8880698d705","_number":738687,"virtual_id_number":738687,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-01 03:26:36.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-09 05:11:06.000000000","post_submit":true,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-08 14:15:04.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2020-07-01 01:10:35.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-07-09 04:35:59.000000000","post_submit":true,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-08 18:21:23.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-07-08 14:02:50.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 23:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-01 04:44:32.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-07-08 14:48:56.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-08 22:53:37.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-07-01 01:27:54.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-01 01:10:39.000000000","_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-01 04:49:44.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-08 13:10:52.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"date":"2020-07-08 18:49:44.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-08 22:52:53.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-07-01 14:18:41.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"date":"2020-07-03 14:17:06.000000000","_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"date":"2020-07-01 01:20:49.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-07-08 14:06:50.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"date":"2020-07-08 22:51:01.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-03 05:57:45.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2020-07-03 05:46:39.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2020-07-08 16:34:41.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-08 22:17:32.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-01 06:11:38.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-08 14:41:06.000000000","_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-02 16:13:05.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-01 09:42:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":2,"date":"2020-07-02 16:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-07 12:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-07 12:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-01 01:10:39.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 01:20:49.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-07-01 01:27:54.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 03:26:36.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 04:44:32.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-07-01 04:49:44.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 06:11:38.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 09:42:29.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-07-01 14:18:41.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-07-02 16:13:05.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-02 16:13:38.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-07-03 05:46:39.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-07-03 05:57:45.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-07-03 14:17:06.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-07-07 12:40:36.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2020-07-08 13:10:52.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-07-08 14:02:50.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 14:06:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-07-08 14:15:04.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-07-08 14:41:06.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 14:48:56.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 16:34:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 18:21:23.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 18:49:44.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 22:17:32.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 22:51:01.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 22:52:53.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-07-08 22:53:37.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-07-08 23:38:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-09 04:35:59.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-09 05:11:06.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75e8536e254e3ba7a277ca8780dfe59408fc0727","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 20:02:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4abdd2cb8da54b4208cdcba37008c20dd247861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 20:02:50.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/cinder/.zuul.yaml@master\", line 34, column 19","accounts_in_message":[],"_revision_number":1},{"id":"97dbbd45d338b216325588bab9238bb6f324068a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-30 20:03:03.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/cinder/.zuul.yaml@master\", line 34, column 19","accounts_in_message":[],"_revision_number":1},{"id":"c856bfd6cbaf35db4e5e97e84f8318f42bed2a62","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-30 20:03:11.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/cinder/.zuul.yaml@master\", line 34, column 19","accounts_in_message":[],"_revision_number":1},{"id":"9cb750606609b7893b45bf55027169cfc5611b92","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-30 20:03:19.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/cinder/.zuul.yaml@master\", line 34, column 19","accounts_in_message":[],"_revision_number":1},{"id":"15834e1ab334345fe90a998e1db373cb5630a55d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 20:04:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d232c225d0ce35e28d1eb2a85a405876dbac0e5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 20:07:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b4bc37b23dd0073bc66b2d243e4d2dac985ce56e","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-06-30 20:21:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/250c20158040809a9ca5e22f16949776899d95bd/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3a31d67a38b9c236f2308599dabbd1919cf2a860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 20:24:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- cinder-tox-coverage https://zuul.opendev.org/t/openstack/build/6f1eef55bd9943c6b16c42c8dae705b8 : SUCCESS in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e077176aa68dd88e34df2a1902d88e1844b4d207","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-30 20:24:56.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/87/738687/3/check/DellEMC_XtremIO_FC/ef63e0d/DellEMC_XtremIO_FC/None : EXCEPTION\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/87/738687/3/check/DellEMC_XtremIO_iSCSI/58dbee6/DellEMC_XtremIO_iSCSI/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"43dd009c0cf81107a121721922d29ea0e7a89bd5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 20:37:44.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4944455f338eee0d88b7ea2a6d9ccdcf451b69fa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 20:42:06.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/87/738687/3/check-not-vote/ext-cinder-zuul/2259781 : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":3},{"id":"51913770ba391175557c371fd843c7474523a1f4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-30 20:47:44.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/87/738687/3/check/dsvm-tempest-cisco-zonemanager-job/b5f2fa5/ : FAILURE in 33m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7b482c5c0629cb7259b708b85b1ec7c889b13268","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-30 20:52:12.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/87/738687/3/check/cinder-storpool-tempest/19e8fc6/ : RETRY_LIMIT in 12m 01s","accounts_in_message":[],"_revision_number":3},{"id":"03003559f45f82020c58fb0eb76799e356c4043f","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-30 20:53:35.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f893ba27635f1684e95f6195c106ba24e93abfb8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 20:55:03.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f52e4e07bb6535f1c6ac95c69ca22fa5e269ea5b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-30 20:56:30.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"74f085e0967283c40949a7a086823d882ae49f6b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-30 21:05:24.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"eeecfad4e82a46c32ec818ccc0963789e44c4a6a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-30 21:10:43.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/738687/3/iscsi/ : SUCCESS in 57m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/738687/3/smb/ : FAILURE in 1h 02m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ba62fae0d04976caf487cad9e74cef31d51bc65f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 21:15:19.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"e5f989477a06652c0e225debd17050e71b96f580","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-06-30 21:23:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/87/738687/3/check/synology-dsm-cinder-iscsi/644e368 : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":3},{"id":"9865a5f27799e5a613b322b3302d18df9b2c51e5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-30 21:31:20.000000000","message":"Patch Set 3:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-87-738687-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"b393282513648ad47aa80250baa18c80be2bf880","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 21:36:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ff596939b928758e030436a4241fb19e3a3151c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:37:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- cinder-tox-coverage https://zuul.opendev.org/t/openstack/build/f1862285e1b049999317ca39791fc095 : ERROR Unable to find playbook /var/lib/zuul/builds/f1862285e1b049999317ca39791fc095/untrusted/project_2/opendev.org/openstack/cinder/playbooks/tox/cover-post.yaml in 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e091929f276893bda5cc2d815d27a325428f5b6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 21:53:54.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"d0c05081d5ddc333b720b76fe46ea7f7643c31ac","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-30 21:56:28.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-738687-4/ns5_nfs : FAILURE in 19m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-738687-4/ns5_iscsi : FAILURE in 19m 26s","accounts_in_message":[],"_revision_number":4},{"id":"e15bc79c55be2128ac9439943a1705035d5bf776","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 22:06:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cff7faa29cd5e5788041c2dc4245242d6863715a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 22:19:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- cinder-tox-coverage https://zuul.opendev.org/t/openstack/build/3f12e8af958c4b6a9fa9911156c6c505 : SUCCESS in 12m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"775d923c23882a9a54c750e263e7849ce1fc0631","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-30 22:27:48.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"f826a8232287b08a20f7724d69f9babd682229b6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 22:34:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"82c967b4a94d87d41dee749b2fdbadcb9031c1a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 22:34:30.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 85, column 31","accounts_in_message":[],"_revision_number":6},{"id":"2c126cd241d3a19d2300a726998722b1e633db42","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-06-30 22:34:45.000000000","message":"Patch Set 6:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 85, column 31","accounts_in_message":[],"_revision_number":6},{"id":"ebe701c6fc2b4fe469a571916ed02d634881dec8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-06-30 22:34:54.000000000","message":"Patch Set 6:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias \u0027gate-irrelevant-files\u0027\n  in \"openstack/cinder/.zuul.yaml@master\", line 85, column 31","accounts_in_message":[],"_revision_number":6},{"id":"328099d6e7283f0123236c8e8d994ed3b4d6995a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-30 22:35:01.000000000","message":"Patch Set 6:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias \u0027gate-irrelevant-files\u0027\n  in \"openstack/cinder/.zuul.yaml@master\", line 85, column 31","accounts_in_message":[],"_revision_number":6},{"id":"d11e22bc8ecfb2f6dd92aa56cb18a3e63361342d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-30 22:35:46.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-738687-3/ns5_nfs : SUCCESS in 2h 24m 28s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-738687-3/ns5_iscsi : SUCCESS in 2h 27m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9480af7b9d1ce4850f4ee7df53c002e08544be86","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 22:36:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"047d33af400b134e4b2c44ccddda9d5d360bd4cf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-30 22:44:36.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-87-738687-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"af5678d6c9f470405face779307c10e2908d1ef6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-30 22:46:47.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-738687-1/ns5_nfs : SUCCESS in 2h 43m 50s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-738687-1/ns5_iscsi : SUCCESS in 2h 21m 57s","accounts_in_message":[],"_revision_number":1},{"id":"de0dcd97f299900ecd71e58386d4dec3a7a43a72","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-30 22:48:20.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-738687-2/ns5_nfs : SUCCESS in 2h 24m 40s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-738687-2/ns5_iscsi : SUCCESS in 2h 43m 49s","accounts_in_message":[],"_revision_number":2},{"id":"88dd124a678ead17308a545ce01d0da71363cf0b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-30 22:57:20.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"9a7dead9459a7f2c6a232c8963efe6e337b34c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 22:58:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- cinder-tox-coverage https://zuul.opendev.org/t/openstack/build/aec3f25610014c5f88d74e9ec1e207e8 : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e0501d2e6ed139c5df0b45b0672ec8f5192825ea","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 22:59:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e635c3e85f09d087d2a55f6e1cc99517f84a509d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 23:04:14.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"053b309d588fb94143c6cd1986bd9f49a9beb87e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 23:19:26.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"dd9ec3765c4d7109d27d3ab4a69de00e9e756fa9","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-30 23:25:19.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"8c2a9fc9e677e3b8812a0ed19a3fd8d21e1901e4","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-06-30 23:28:45.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"dd446567102e28718b03ddb0975196cbbe5d539b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-30 23:28:50.000000000","message":"Patch Set 5:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-87-738687-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"f4898182a8a92b7bc05743056df46f8306ef4688","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 23:33:36.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/87/738687/8/check-not-vote/ext-cinder-zuul/ea7380a : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":8},{"id":"7659171b3fe7019e9788d1372afbefd1f992bca9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-06-30 23:57:00.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"c78eb75fa2765b67ea25fdf2529b55a74dc5d9e8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-06-30 23:59:49.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/738687/8/iscsi/ : SUCCESS in 1h 00m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/738687/8/smb/ : FAILURE in 54m 08s","accounts_in_message":[],"_revision_number":8},{"id":"c59feb66cecdb7548ecb8057b7dba92eb494c0a5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-01 00:12:47.000000000","message":"Patch Set 7:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-87-738687-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"2b1e3c1228216efe08b29efe685f37789ee62d1c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-01 00:16:20.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"daf78c918fd02fab81fd0e499c080a22d1dd7a8a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-01 00:18:01.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/87/738687/8/check/dsvm-tempest-cisco-zonemanager-job/6e49e62/ : FAILURE in 34m 40s","accounts_in_message":[],"_revision_number":8},{"id":"fc3599fc2031172f7b38069628800d7cfbfbe96e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 00:20:08.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"5fd743bceedfe6c8a9d00ac12062202eba59b61d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 00:36:19.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"5c7c0272221d7218e9436f104a244071ce9ba100","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-07-01 00:48:58.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/87/738687/8/check/cinder-storpool-tempest/be2e35c/ : RETRY_LIMIT in 11m 44s","accounts_in_message":[],"_revision_number":8},{"id":"cdcf894c57b75e7b99c8f8b3f89be302e6d54f1c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-01 00:49:31.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-738687-5/ns5_nfs : SUCCESS in 2h 26m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-738687-5/ns5_iscsi : SUCCESS in 2h 43m 13s","accounts_in_message":[],"_revision_number":5},{"id":"9f0a72df6c33d84ca9f6e882868552fab8b8fbe5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 00:55:28.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"2c7f83b9847f8aa3459074c86e548f4af48c6e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 01:01:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0adbf77dbf8468b8329bdac89c12524 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ac123b80734b8cb1638ca7122c1cf8 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/102c6f8146b64045954a943da70f4b76 : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eab5f42bc5bf44efb4ef6e63a8f190d8 : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb4a91d8a914dfdab7e94a00749b600 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/18cb1a34a9654b118f525d9200fb3c98 : SUCCESS in 1h 05m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dcdc1bce54894668aeadb22a0f02f92d : SUCCESS in 1h 20m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/91fb53dafce24ccbb78e30718cc90703 : SUCCESS in 12m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3a1fe4d21a7454db0cdc74b1175a2a1 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92572c37adb9436980e0fde07a069ed6 : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6338f4c7783941758ec3d3cb9939c7e8 : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3a0e1635e14649488a2acec7f555d134 : SUCCESS in 42m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b635e090720a46539198e87216a9fb20 : SUCCESS in 4m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f7f5bfbe14034775b4fba1da12bd5884 : FAILURE in 26m 25s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d4012b9a9dd946139d6e4a988213c11c : SUCCESS in 1h 19m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4a23d14bb9a34e51962154c12a4d080d : SUCCESS in 1h 12m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6cf0992e243b457998cfe1eb0b5e76be : FAILURE in 1h 44m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cc39a1e52f8a4acf9b7eabe8e9a19a59 : SUCCESS in 1h 19m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97f5e4cc9999484b8e9137345b963608 : SUCCESS in 1h 57m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3832520b997d46639b9f8209751483b8 : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":8},{"id":"1eb2dc15fdd5a12b6c189be7eec512973ab44cf3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-01 01:10:35.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"d2b0bee98e7d7963e1cff27b32d3d1c49b36f136","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-01 01:10:39.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/87/738687/8/check/nimble-iscsi-driver-dsvm-volume/67c2dfc : SUCCESS in 2h 11m 18s\n- nimble-fc-driver-dsvm-volume https://ftp.nimblestorage.com/openstack_ci/iscsi/87/738687/8/check/nimble-fc-driver-dsvm-volume/1342726 : FAILURE in 43m 05s","accounts_in_message":[],"_revision_number":8},{"id":"67b8e9afdca9321d161c789040c6b218c9982456","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 01:16:43.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/72678f839e6e4af2b3799a97bc987b0e9086c920/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"29b3080adb8b66baadc053ad5973617bf7d55623","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-01 01:18:46.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-738687-7/ns5_nfs : SUCCESS in 2h 41m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-738687-7/ns5_iscsi : SUCCESS in 2h 39m 18s","accounts_in_message":[],"_revision_number":7},{"id":"e103274d5560099e2e26643cffc2f948740fc04d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-07-01 01:20:49.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-fc http://openstack-ci-logs.aws.infinidat.com/87/738687/8/check/dsvm-tempest-infinibox-fc/657db6c : FAILURE in 1h 19m 07s\n- dsvm-tempest-infinibox-iscsi http://openstack-ci-logs.aws.infinidat.com/87/738687/8/check/dsvm-tempest-infinibox-iscsi/c64ef17 : FAILURE in 1h 02m 07s","accounts_in_message":[],"_revision_number":8},{"id":"525f3ac8d1e124cbe53ab6cd2917f623eeef71ba","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-01 01:21:21.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-738687-6/ns5_nfs : SUCCESS in 2h 26m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-738687-6/ns5_iscsi : SUCCESS in 2h 46m 44s","accounts_in_message":[],"_revision_number":6},{"id":"f2eb7e836008dce4d62291714203351a2388d5ed","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-01 01:27:54.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-738687-8/ns5_nfs : SUCCESS in 2h 24m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-738687-8/ns5_iscsi : SUCCESS in 2h 28m 27s","accounts_in_message":[],"_revision_number":8},{"id":"8c215e6256a84e04ff7af424bef9fd73e9606193","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 01:33:35.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"6fe7ba301f851ff3a48c9278a0d528606732799f","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 01:34:01.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/4a5be46398e6ae274e332726d153929c901c15c0/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"97b710ef490d9b30bbe7a426017016a281a2fa8d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-01 01:44:35.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/87/738687/8/check-not-vote/ext-cinder-zuul/c13d85f : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":8},{"id":"fee8514822aff49d8e41264afbbdfdd4f621174e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 01:48:36.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"e458188566921146ee80f58224d23f641f584fdf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-01 01:50:22.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/87/738687/8/check/dsvm-tempest-cisco-zonemanager-job/c717221/ : FAILURE in 34m 37s","accounts_in_message":[],"_revision_number":8},{"id":"2655346682af79f67bf4de624aff9726712c0437","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-07-01 01:54:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/87/738687/8/check/cinder-storpool-tempest/ccc4321/ : RETRY_LIMIT in 12m 02s","accounts_in_message":[],"_revision_number":8},{"id":"e80736f40b56d45259546db778cf01a7362c5d14","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-01 02:01:26.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"1cef098b801f9bb99508dfbfed4ce42c7b09be5f","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 02:09:23.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/a3326943b8c1bb80396b045dd1b39508f2f00d45/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"bb7a2bceca23483f9cbb1d9c4fa2019d69faf2f2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-01 02:12:41.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/738687/8/iscsi/ : SUCCESS in 1h 01m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/738687/8/smb/ : FAILURE in 53m 03s","accounts_in_message":[],"_revision_number":8},{"id":"866b11cc66930c760d66097aaa155c4c07314537","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 02:25:58.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"4ee09406b4193fcc9a1f5172eeb8903a17d0feaa","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 02:28:08.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/e33aeda2d83f3c103d896c4c3194b2b42d690b80/index.html : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"b931cf51dd21bae12b85ec01bc14f39a10e00277","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 02:46:11.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/a3055860d274791b2118e8a901ed9201e71bdb00/index.html : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"e24ce97e88f1e5eed6ae35a988e26d31c401e703","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-01 02:55:25.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/87/738687/8/check/tempest-dsvm-full-bionic/fa31cab/ : SUCCESS in 1h 33m 03s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/87/738687/8/check/devstack-plugin-ceph-dsvm-bionic/d757b5c/ : SUCCESS in 1h 33m 16s","accounts_in_message":[],"_revision_number":8},{"id":"cb35d4dd84ea6c39e2c2990760ed34a320d1e82b","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 03:07:10.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/d40886018e44f5e7362946de0f75bfbaea0d45b8/index.html : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"3d7c524252c9db339a810c393a006711d6df59a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 03:13:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21409b37f9bf4f11b1233aaafed5c742 : SUCCESS in 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8d3fda69ee406ca7ea6eaf4e8f164f : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1743c3e822b84d7db65113ccd3391a65 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e3b9853e9584c88a3f6d4b7ff161354 : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b029c971ec0b416bac09282e1cc96d16 : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/17321663a24f4caeab8f648399b08040 : SUCCESS in 1h 04m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21187c0a7baf4bd6acdbf8ffccef5614 : SUCCESS in 1h 05m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7bb5c355414a40b28caffafae22ad2f6 : SUCCESS in 12m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d0a4344f7714ba39055f15c12d2c392 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a079ec49e3e45169d24621c985673c6 : SUCCESS in 5m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/86c48b9c60744d558e7cb04a4988761d : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e37698297aaa412588c99434e179a34a : SUCCESS in 43m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f2e2808aa6b4b68a90aa0a98b2aa289 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/928b4bdcf57e445689df84189111023b : SUCCESS in 59m 23s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f8652411261b437b826e1a964b9b88aa : SUCCESS in 1h 27m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6dc3733869134858930eea83cba02989 : SUCCESS in 1h 11m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9767aad3d92e43d7babf98494bd1fc99 : SUCCESS in 1h 16m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/33b4d8225c274d8f93e88193350a8490 : SUCCESS in 1h 20m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0fbb937d444f4edc8c74c534295f2176 : SUCCESS in 1h 57m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37f24fe096c44901aff685c61964c92a : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":8},{"id":"9af6c6cca9afb451ce8ff604c5ef9771a54eef2c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-01 03:20:34.000000000","message":"Patch Set 8:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-87-738687-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"88b852a7657a9fa8dab0f81e80d0edfafcaa4d47","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 03:26:36.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/416e61e6499433f5b535d7011a6b23e26b31989d/index.html : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"68e286e1a2e816a063f3d72907efda3c87c77c28","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 03:30:51.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"50e76d73ae3eb21672595469ec7076f790aec544","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-01 04:11:39.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/87/738687/8/check/synology-dsm-cinder-iscsi/c26bacb : FAILURE in 57m 32s","accounts_in_message":[],"_revision_number":8},{"id":"f577a65a940872f636c87ad6ae9afeadce9ae146","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-01 04:41:08.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/87/738687/8/upstream-solidfire-check/cinder-solidfire-iSCSI/22065c4/ : SUCCESS in 1h 26m 23s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"afc7fc3307c77f4f34c29b9baf38eab76452852a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-07-01 04:44:32.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/87/738687/8/check/DellEMC_SC_FC/24f0d50/DellEMC_SC_FC/145 : SUCCESS in 1h 18m 11s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/87/738687/8/check/DellEMC_SC_iSCSI/3744194/DellEMC_SC_iSCSI/125 : SUCCESS in 1h 30m 42s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"be78af8a144b82083f7eb62c9a30bdfe3a5f7f3b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 04:46:54.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"1c8c59da62773fa3e07cd1da4381a57f9b11d401","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-01 04:49:44.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/87/738687/8/upstream-check/cinder-cDOT-iSCSI/db2551e/ : FAILURE in 1h 35m 17s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/87/738687/8/upstream-check/cinder-cDOT-iSCSI-IPv6/a2b28ab/ : SUCCESS in 1h 21m 51s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/87/738687/8/upstream-check/cinder-cDOT-NFS/4cb8bc0/ : SUCCESS in 34m 20s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/87/738687/8/upstream-check/cinder-cDOT-NFS-IPv6/b2d13f5/ : SUCCESS in 1h 22m 31s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/87/738687/8/upstream-check/cinder-cDOT-FCP/dfa41da/ : SUCCESS in 52m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"5a9f9a320b6a6090afeeee65a324de512acb21c8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-07-01 06:11:38.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/87/738687/8/check/DellEMC_PowerMAX_FC/63eafca/DellEMC_PowerMAX_FC/181 : SUCCESS in 4h 19m 36s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/87/738687/8/check/DellEMC_PowerMAX_iSCSI/ffc0c30/DellEMC_PowerMAX_iSCSI/157 : SUCCESS in 5h 51m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"1a5e40b026a77e965be524a5a1e81aed77731be2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-01 06:47:46.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"9d575e2eb18d6bb8dadcd9f3935dd7d5fbe6cd1a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-01 09:42:29.000000000","message":"Patch Set 8: Code-Review+1\n\nCurrent coverage is 78%, so I would like to set a threshold about 75% and make this job voting. I know that somebody will not agree with me, so just +1 for this patch","accounts_in_message":[],"_revision_number":8},{"id":"62ae7a458dc20040fa465d032058e1cf08e24bcc","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-07-01 14:18:41.000000000","message":"Patch Set 8:\n\nI see some of the files have very less coverage ~40-50% and the final percentage is an average? 78%\nNot sure but the final average doesn\u0027t seem like a good measure for checking coverage, coverage w.r.t. files still seems reasonable.","accounts_in_message":[],"_revision_number":8},{"id":"7fae738f4bbd4f0b9aed7a0c7975c7da5bbe6cc8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-02 16:13:05.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/87/738687/8/20200702223347/NEC-FC : SUCCESS in 2h 22m 54s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/87/738687/8/20200702223347/NEC-ISCSI : SUCCESS in 2h 39m 16s","accounts_in_message":[],"_revision_number":8},{"id":"2274cf64e4e2325e2e5708afae09243d8f67daa3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-07-02 16:13:38.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"102a4fc51c6f10336ed385da78b27a24a50ce0d5","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-03 00:25:35.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/738687-3/HPMSA-iSCSI : FAILURE in 17:40.90\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/738687-3/HPMSA-FC : FAILURE in 17:36.78","accounts_in_message":[],"_revision_number":3},{"id":"5051a4ed4cf0c9c272c38c227192b3d0d13f81f1","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-03 01:29:05.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/738687-3/Lenovo-iSCSI : FAILURE in 15:24.86\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/738687-3/Lenovo-FC : FAILURE in 15:20.34","accounts_in_message":[],"_revision_number":3},{"id":"6b24179f180af88f3f379d2a341a5904a429825a","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-03 01:32:36.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/738687-4/HPMSA-iSCSI : FAILURE in 17:24.34\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/738687-4/HPMSA-FC : FAILURE in 17:10.30","accounts_in_message":[],"_revision_number":4},{"id":"18248863fd5552c85c79e5785257fa3ca0be810b","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-03 02:35:17.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/738687-4/Lenovo-iSCSI : FAILURE in 17:07.97\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/738687-4/Lenovo-FC : FAILURE in 17:23.36","accounts_in_message":[],"_revision_number":4},{"id":"715cd0d7f765dd4ee656f92f39e5e155660106bb","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-03 03:42:30.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/738687-7/Lenovo-iSCSI : FAILURE in 15:48.00\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/738687-7/Lenovo-FC : FAILURE in 15:55.88","accounts_in_message":[],"_revision_number":7},{"id":"60a64d635c903c5035ae4083e71a0e069a96d10e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-03 03:44:07.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/738687-7/HPMSA-iSCSI : FAILURE in 17:39.41\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/738687-7/HPMSA-FC : FAILURE in 17:48.84","accounts_in_message":[],"_revision_number":7},{"id":"6ec14b2a5a955da7a0830404f6c4c36307bea7fb","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-03 04:00:16.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/738687-3/Seagate-iSCSI : SUCCESS in 46:47.52\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/738687-3/Seagate-FC : SUCCESS in 48:16.80","accounts_in_message":[],"_revision_number":3},{"id":"7574bbf41ed91cdca3b6a14e93293c3be2cb0abc","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-03 05:46:39.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/738687-8/Lenovo-iSCSI : FAILURE in 15:25.59\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/738687-8/Lenovo-FC : FAILURE in 15:29.45","accounts_in_message":[],"_revision_number":8},{"id":"4e053bd308b2f6b9673f52a542729ce95b15a49c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-03 05:51:33.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/738687-4/Seagate-iSCSI : SUCCESS in 41:35.65\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/738687-4/Seagate-FC : SUCCESS in 38:32.81","accounts_in_message":[],"_revision_number":4},{"id":"d7b2d735100e13abc0562a3d03f1820812d87f87","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-03 05:57:45.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/738687-8/HPMSA-iSCSI : FAILURE in 17:36.94\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/738687-8/HPMSA-FC : FAILURE in 17:26.56","accounts_in_message":[],"_revision_number":8},{"id":"c0b6742c1c16c3045d02a23c0dcc07d7a2a682ee","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-03 07:52:36.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/738687-5/Seagate-iSCSI : FAILURE in 47:21.51\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/738687-5/Seagate-FC : SUCCESS in 40:32.79","accounts_in_message":[],"_revision_number":5},{"id":"fcdc53062fe3835c5dc8dff7175ddf0afb96b8cd","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-03 10:09:18.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/738687-7/Seagate-iSCSI : SUCCESS in 51:39.34\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/738687-7/Seagate-FC : SUCCESS in 50:16.95","accounts_in_message":[],"_revision_number":7},{"id":"a95f6cdf9801f2603709289a0c12b901f1c8fe5c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-03 14:17:06.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/738687-8/Seagate-iSCSI : SUCCESS in 45:01.59\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/738687-8/Seagate-FC : SUCCESS in 43:39.91","accounts_in_message":[],"_revision_number":8},{"id":"8d48e5f00feeed86a9d646250a6bbeb05343f76d","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-07-07 12:40:36.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"503eb7bb5136f1581d3485ea52254d4b703a8f81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 12:40:53.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6a144fc0d73c4adc0ba7f130aa82bdf65a51cee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 14:16:53.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04c3707bf7e44280b29c7214aba56b3e : FAILURE in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02b97e803fc7460dbab28d67220c2bb6 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd5c73dec1146b5a9c7378c338633aa : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44643f013aa344968b80ef8c5171d1c9 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7212f4c1286405f8644415abfe918f0 : SUCCESS in 16m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/96c130a4578c49039cb4c2ce56755cc8 : SUCCESS in 1h 00m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a6d0e2a2e0940088c1e6523ce08352a : SUCCESS in 1h 08m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/761c8ba31ff843a793c0f2b7d5f11c4f : SUCCESS in 1h 21m 37s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/28f9aee7e5bd475cabef55c039b3a467 : SUCCESS in 1h 32m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e5d526415b744a88c96a30039e09aff : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":8},{"id":"5da0b2d64ad04cf31bd6ba48b8f3ea7fb793b93e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-07-07 14:49:54.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"58d461164c1953abcc7e5601bfd6c02be07e471b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-07 15:15:33.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/87/738687/8/check/dsvm-tempest-zadara-driver/f90411c : FAILURE in 25m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bd90c562f85b5ec253f5c4215d74bc913d8a8e54","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-07 15:16:02.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/87/738687/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/5af362e : FAILURE in 25m 50s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/87/738687/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/7b7aeef : FAILURE in 20m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"0620479a118fc3aeaff2650c654211355f7ef047","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-07 15:22:42.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/87/738687/8/check-not-vote/ext-cinder-zuul/7a9cad7 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":8},{"id":"3f972350571dc388f93fb39a2d7a72ce34308393","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-07 15:29:13.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/87/738687/8/check/dsvm-tempest-cisco-zonemanager-job/7177a2f/ : FAILURE in 34m 04s","accounts_in_message":[],"_revision_number":8},{"id":"d4da69e7ff01089d76d2bc6aa3bc39831a6a59ee","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-07 15:35:14.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"9a99f33521253bbb41f60cc64a37e3d098dd2807","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-07 15:41:21.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"3239d1aeb40dfc883de64fb4096fa271e59b3088","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-07 15:52:15.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/738687/8/iscsi/ : SUCCESS in 1h 01m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/738687/8/smb/ : FAILURE in 53m 35s","accounts_in_message":[],"_revision_number":8},{"id":"bb9716cfc5c9a54eb64bf99036d147a2ae7c3d6b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-07-07 18:15:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/87/738687/8/check/cinder-storpool-tempest/4d4d5eb/ : RETRY_LIMIT in 11m 20s","accounts_in_message":[],"_revision_number":8},{"id":"132510cbde683a5139930f6dc4bd141f0baf313e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-07 19:21:21.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"8d372ce54c197eac844c8a0f0c202360ae89676f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-07 20:02:45.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic tempest-dsvm-full-bionic : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/87/738687/8/check/devstack-plugin-ceph-dsvm-bionic/a18207e/ : SUCCESS in 1h 43m 39s","accounts_in_message":[],"_revision_number":8},{"id":"e0df2fd6f80df9f5d6aad66e6ff2f8ea830f1dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 21:03:38.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1e50d9177c14bb5b21c6828066b94e0 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e75ededc9b4648ad3278eaa11f81c0 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dea4ff152c254c4f99aaab9ed37d4e1e : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c833ea7387344686b1ab89401eb78aeb : SUCCESS in 14m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b3416526e7d424396b961dd7c82fd4a : SUCCESS in 15m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/9938e59b4d0d40939991ac98fe1d947e : SUCCESS in 58m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/557f2f4b84f24b349909e66c417fcb87 : SUCCESS in 1h 06m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/33d6eaf845164fe088bcd83ccba27ef2 : SUCCESS in 14m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/75bc6a5b22e4496bb1f828aca5b03084 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0b5b895957141bfa14b8dd727abd438 : SUCCESS in 7m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8f7394743c3f40f7a12e427a45fc6f76 : SUCCESS in 7m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2cdea73c7a3a4bd69b42d4b5c75ea6dc : SUCCESS in 39m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3094bbbb5a43415d88ba135e9497da9d : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c4ab06d7c724eeda796363dc6a36695 : SUCCESS in 1h 05m 59s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/977f467ff8974b7ba24cf5c411768ebb : FAILURE in 1h 36m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/635e204e03b84030abda48c8fa9418e1 : SUCCESS in 1h 16m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/efbf5da0473141e28931ceb62a453cb5 : FAILURE in 1h 21m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/727697fbb464445a9fcbfdb44a428dc2 : RETRY_LIMIT in 34m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2c25121dd484b4293920ac401151cfd : SUCCESS in 1h 31m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/240de643035f4a7dbae3e8e0f1531a5f : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":8},{"id":"37784bfe2f47d11a4e45e405a5f292840799daa5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-08 00:05:16.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"3a1fda735cfee30c019cd24201245cd936c69b0a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-08 13:10:52.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"9388ce9d100a51029c090301b8ca0af0ed05c5cd","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-08 14:02:50.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"f00c6d6abc26f155220aa0f82d284b043344be25","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-08 14:06:50.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/738687/8/iscsi/ : SUCCESS in 55m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/738687/8/smb/ : FAILURE in 53m 04s","accounts_in_message":[],"_revision_number":8},{"id":"ea9653f49de7d76c29e7d78327a1edc848fa6ee1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-08 14:15:04.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/87/738687/8/check/dsvm-tempest-cisco-zonemanager-job/b9212a2/ : FAILURE in 33m 54s","accounts_in_message":[],"_revision_number":8},{"id":"8b88161e5199aadf17e5eab948c6e61cb918e12f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-07-08 14:41:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/87/738687/8/check/cinder-storpool-tempest/2ad7e8f/ : RETRY_LIMIT in 12m 01s","accounts_in_message":[],"_revision_number":8},{"id":"1973699ebfc8e2cd55fd79472f9dd63080295da7","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-08 14:48:56.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/87/738687/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/cd4f5d8 : FAILURE in 20m 12s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/87/738687/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/139cdb8 : FAILURE in 20m 38s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"0e11b51cbe7b8de49a264f77a091633aa5b7fc5f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-08 14:53:36.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"c5b6163900ff0e2ce06bff93ebd1b7761cadc994","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-08 16:34:41.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/87/738687/8/check-not-vote/ext-cinder-zuul/dc21bc6 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":8},{"id":"bc69ca81a1510e65c6aa94bb256ed0a79aea215a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-08 18:21:23.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/87/738687/8/check/tempest-dsvm-full-bionic/c6660f7/ : SUCCESS in 1h 39m 44s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/87/738687/8/check/devstack-plugin-ceph-dsvm-bionic/f404ef4/ : SUCCESS in 1h 30m 20s","accounts_in_message":[],"_revision_number":8},{"id":"9a0034e7dad49575ffeaf0d69b8d4106812a8647","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-08 18:44:44.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"aaafe919ee3993e443b501c3fd69df6d9f0e822b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-08 18:49:44.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/87/738687/8/check/dsvm-tempest-zadara-driver/61b64eb : SUCCESS in 1h 09m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"681304edab4349b669ea77f9617a2a0de92bb756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 21:51:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9efb8e4229bc47fa814b62b6f5e6c9df : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19da75c463c49dba8ced8f1b9ed2d6c : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7647a034e49f489dbd9cbcbca3ac42a5 : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b56149d87346a4a67889373d7e8794 : SUCCESS in 15m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1bde2ee17e458ab7cba9f36ae8fef6 : SUCCESS in 15m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/47ec1c188bfb42c39d8b05a7c990f614 : SUCCESS in 1h 01m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85e124dc1fd4494095baa0f2bee5d431 : SUCCESS in 1h 04m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cece194b3e2649f599ac6009491fb344 : SUCCESS in 11m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf608a8c81ef4c77bedfd4cb6b0c25f9 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/01e19924dfa946afaad20cef1f747c0e : SUCCESS in 9m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7afcff09fd06488891361e0f71c0aa3c : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ebf447b4a9e420694c6ffaeeb904551 : SUCCESS in 38m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/24d95c0c9d03451faee8698887950dec : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a9980daf6ec745e4811a79d73df46aa5 : SUCCESS in 1h 08m 25s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e63846dbf63a40d094d7d0a87e1ba55d : SUCCESS in 1h 20m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/39ae78f4ffee4331a580c8cbc2018fb2 : SUCCESS in 1h 15m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6178938120504cae823f081a97f693bb : SUCCESS in 1h 31m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dcc1eff7604940679af7e1d8e87a2773 : FAILURE in 2h 12m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2949b6a827a4462b396c0c7128db3a6 : SUCCESS in 1h 58m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84b68920127b497bb88f99bb2d6f4b64 : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":8},{"id":"d563356aad94b28f9f36a717e9b9f66f531e8836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 21:52:01.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2e4304398ba272cf564799982930054d42a4c19a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-08 22:17:32.000000000","message":"Patch Set 8:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-87-738687-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"139a3f912e0aafab35d1a1bf7c598cc5a493ce71","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-08 22:51:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/87/738687/8/check/synology-dsm-cinder-iscsi/70f41a4 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":8},{"id":"108b3e114e126abacc42206572124c530e04a2c7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-08 22:52:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/87/738687/8/upstream-solidfire-check/cinder-solidfire-iSCSI/7ba3427/ : SUCCESS in 1h 00m 24s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"8e43b22fa3c411454fb62325f4f3fc0a2c50a75c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-08 22:53:37.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/87/738687/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"78030fe2f40fe9a3180d1a2442cc7dbf72cdf693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 23:38:35.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59b6ffa77afb4f20a4cf48918ed4fff3 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640928b0d82f413db571fedfb083f9f0 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1375ca196b7b46c7a6a22fd4ddf5f457 : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2362b0db33f0404dbd9eb5052b224849 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac44c0ba44284bedbcf38f25c22b15a0 : SUCCESS in 14m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/a316f116a3e34596b59d287657e06c03 : SUCCESS in 1h 09m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce6168651c8b46ec956cdbb4e67df3d0 : SUCCESS in 1h 13m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/68292837b92b4e69b79e4e939ef466e9 : SUCCESS in 1h 14m 42s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91c8fdcf05e34377a6ffd06fa238b100 : SUCCESS in 1h 39m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86f310f12a1c4f9783965b5dd866d051 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":8},{"id":"7adf90fa811c4596f6eac879258b59aee0f4e948","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 23:38:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"9a2ffa7643689b03a18c3865747bb8ca9224e95f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 23:43:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787df7b18d2e49b4ad486d7734e9ba2c : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":8},{"id":"39b124c7e7ceec9de9a7dcbcc370b94f3b2b4796","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-09 00:01:23.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/738687-8/Lenovo-iSCSI : FAILURE in 46:01.27\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/738687-8/Lenovo-FC : SUCCESS in 43:34.33","accounts_in_message":[],"_revision_number":8},{"id":"357004759eed699884ba1818f77a051e1ecbb70e","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-09 00:08:14.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/738687-8/Seagate-iSCSI : SUCCESS in 46:28.14\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/738687-8/Seagate-FC : SUCCESS in 45:29.37","accounts_in_message":[],"_revision_number":8},{"id":"eb22df4438dcd7c1ddf517404c4354edb5e55fc2","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-09 00:23:51.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/738687-8/HPMSA-iSCSI : FAILURE in 47:35.07\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/738687-8/HPMSA-FC : SUCCESS in 48:28.58","accounts_in_message":[],"_revision_number":8},{"id":"2f62632d40ae6a09e3752dce42fa361577e3133c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-09 04:35:59.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS http://publiclogs.emc.com/87/738687/8/check/DellEMC_VxFlexOS/3cd152c/DellEMC_VxFlexOS/207 : ABORTED in 4h 00m 03s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"a5613c5d37131e34afa136e5e19a79bbb6489210","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-09 05:11:06.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/87/738687/8/check/DellEMC_XtremIO_FC/480ccf5/DellEMC_XtremIO_FC/322 : SUCCESS in 1h 34m 31s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/87/738687/8/check/DellEMC_XtremIO_iSCSI/bec3688/DellEMC_XtremIO_iSCSI/315 : SUCCESS in 1h 36m 00s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"416e61e6499433f5b535d7011a6b23e26b31989d","revisions":{"250c20158040809a9ca5e22f16949776899d95bd":{"kind":"REWORK","_number":1,"created":"2020-06-30 20:02:33.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/738687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/738687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/738687/1"}}},"commit":{"parents":[{"commit":"6c4035ce2c9d3e01acacfd428da4ebcae6cc6114","subject":"Merge \"PowerMax Driver - RDF State Validation Enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c4035ce2c9d3e01acacfd428da4ebcae6cc6114"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"subject":"WIP: Add non-voting coverage job","message":"WIP: Add non-voting coverage job\n\nChange-Id: Icbc1fbb1c9dcf3312860e504adb134811259c817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/250c20158040809a9ca5e22f16949776899d95bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/250c20158040809a9ca5e22f16949776899d95bd"}]},"branch":"refs/heads/master"},"72678f839e6e4af2b3799a97bc987b0e9086c920":{"kind":"REWORK","_number":2,"created":"2020-06-30 20:04:12.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/738687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/738687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/738687/2"}}},"commit":{"parents":[{"commit":"6c4035ce2c9d3e01acacfd428da4ebcae6cc6114","subject":"Merge \"PowerMax Driver - RDF State Validation Enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c4035ce2c9d3e01acacfd428da4ebcae6cc6114"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:04:01.000000000","tz":-240},"subject":"WIP: Add non-voting coverage job","message":"WIP: Add non-voting coverage job\n\nChange-Id: Icbc1fbb1c9dcf3312860e504adb134811259c817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72678f839e6e4af2b3799a97bc987b0e9086c920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72678f839e6e4af2b3799a97bc987b0e9086c920"}]},"branch":"refs/heads/master"},"4a5be46398e6ae274e332726d153929c901c15c0":{"kind":"REWORK","_number":3,"created":"2020-06-30 20:07:49.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/738687/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/738687/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/738687/3"}}},"commit":{"parents":[{"commit":"6c4035ce2c9d3e01acacfd428da4ebcae6cc6114","subject":"Merge \"PowerMax Driver - RDF State Validation Enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c4035ce2c9d3e01acacfd428da4ebcae6cc6114"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:07:38.000000000","tz":-240},"subject":"WIP: Add non-voting coverage job","message":"WIP: Add non-voting coverage job\n\nChange-Id: Icbc1fbb1c9dcf3312860e504adb134811259c817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a5be46398e6ae274e332726d153929c901c15c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a5be46398e6ae274e332726d153929c901c15c0"}]},"branch":"refs/heads/master"},"a3326943b8c1bb80396b045dd1b39508f2f00d45":{"kind":"REWORK","_number":4,"created":"2020-06-30 21:36:38.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/738687/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/738687/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/738687/4"}}},"commit":{"parents":[{"commit":"6c4035ce2c9d3e01acacfd428da4ebcae6cc6114","subject":"Merge \"PowerMax Driver - RDF State Validation Enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c4035ce2c9d3e01acacfd428da4ebcae6cc6114"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 21:36:29.000000000","tz":-240},"subject":"WIP: Add non-voting coverage job","message":"WIP: Add non-voting coverage job\n\nChange-Id: Icbc1fbb1c9dcf3312860e504adb134811259c817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3326943b8c1bb80396b045dd1b39508f2f00d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3326943b8c1bb80396b045dd1b39508f2f00d45"}]},"branch":"refs/heads/master"},"e33aeda2d83f3c103d896c4c3194b2b42d690b80":{"kind":"REWORK","_number":5,"created":"2020-06-30 22:06:00.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/738687/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/738687/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/738687/5"}}},"commit":{"parents":[{"commit":"6c4035ce2c9d3e01acacfd428da4ebcae6cc6114","subject":"Merge \"PowerMax Driver - RDF State Validation Enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c4035ce2c9d3e01acacfd428da4ebcae6cc6114"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 22:05:48.000000000","tz":-240},"subject":"WIP: Add non-voting coverage job","message":"WIP: Add non-voting coverage job\n\nChange-Id: Icbc1fbb1c9dcf3312860e504adb134811259c817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e33aeda2d83f3c103d896c4c3194b2b42d690b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e33aeda2d83f3c103d896c4c3194b2b42d690b80"}]},"branch":"refs/heads/master"},"a3055860d274791b2118e8a901ed9201e71bdb00":{"kind":"REWORK","_number":6,"created":"2020-06-30 22:34:12.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/738687/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/738687/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/738687/6"}}},"commit":{"parents":[{"commit":"6c4035ce2c9d3e01acacfd428da4ebcae6cc6114","subject":"Merge \"PowerMax Driver - RDF State Validation Enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c4035ce2c9d3e01acacfd428da4ebcae6cc6114"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 22:33:53.000000000","tz":-240},"subject":"WIP: Add non-voting coverage job","message":"WIP: Add non-voting coverage job\n\nChange-Id: Icbc1fbb1c9dcf3312860e504adb134811259c817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3055860d274791b2118e8a901ed9201e71bdb00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3055860d274791b2118e8a901ed9201e71bdb00"}]},"branch":"refs/heads/master"},"d40886018e44f5e7362946de0f75bfbaea0d45b8":{"kind":"REWORK","_number":7,"created":"2020-06-30 22:36:50.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/738687/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/738687/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/738687/7"}}},"commit":{"parents":[{"commit":"6c4035ce2c9d3e01acacfd428da4ebcae6cc6114","subject":"Merge \"PowerMax Driver - RDF State Validation Enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c4035ce2c9d3e01acacfd428da4ebcae6cc6114"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 22:36:42.000000000","tz":-240},"subject":"WIP: Add non-voting coverage job","message":"WIP: Add non-voting coverage job\n\nChange-Id: Icbc1fbb1c9dcf3312860e504adb134811259c817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d40886018e44f5e7362946de0f75bfbaea0d45b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d40886018e44f5e7362946de0f75bfbaea0d45b8"}]},"branch":"refs/heads/master"},"416e61e6499433f5b535d7011a6b23e26b31989d":{"kind":"REWORK","_number":8,"created":"2020-06-30 22:59:05.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/738687/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/738687/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/738687/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/738687/8"}}},"commit":{"parents":[{"commit":"6c4035ce2c9d3e01acacfd428da4ebcae6cc6114","subject":"Merge \"PowerMax Driver - RDF State Validation Enhancements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c4035ce2c9d3e01acacfd428da4ebcae6cc6114"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 20:01:51.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 22:58:03.000000000","tz":-240},"subject":"Add non-voting code coverage job","message":"Add non-voting code coverage job\n\nChange-Id: Icbc1fbb1c9dcf3312860e504adb134811259c817\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/416e61e6499433f5b535d7011a6b23e26b31989d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/416e61e6499433f5b535d7011a6b23e26b31989d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
