)]}'
{"id":"openstack%2Fcinder~741158","triplet_id":"openstack%2Fcinder~master~Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6","project":"openstack/cinder","branch":"master","topic":"victoria-drivers","hashtags":[],"change_id":"Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6","subject":"Add Cinder driver for Dell EMC PowerStore","status":"MERGED","created":"2020-07-15 11:23:58.000000000","updated":"2020-07-24 02:52:33.000000000","submitted":"2020-07-24 02:50:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741158-1595559049383-1a0e6018","meta_rev_id":"79da61817a3fa48e1a04a6d5ca2be53eecb415c2","_number":741158,"virtual_id_number":741158,"owner":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-23 03:38:37.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-23 22:05:08.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2020-07-24 01:28:08.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-07-23 22:42:39.000000000","_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"date":"2020-07-24 00:40:15.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 22:45:00.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-07-23 22:18:46.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 02:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 01:29:25.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-07-23 21:55:19.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 02:05:22.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-07-22 21:47:38.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 01:33:56.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 18:27:43.000000000","_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 01:29:49.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-07-23 22:46:15.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"date":"2020-07-23 02:55:40.000000000","_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"date":"2020-07-24 01:47:11.000000000","_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"date":"2020-07-22 20:50:53.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 19:42:08.000000000","_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"date":"2020-07-23 22:28:21.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"date":"2020-07-22 21:12:53.000000000","_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 00:21:03.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 02:03:20.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2020-07-24 02:10:52.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2020-07-23 22:02:20.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 00:20:27.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 20:01:35.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-22 20:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-23 12:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2020-07-22 20:34:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":1,"date":"2020-07-22 19:52:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 02:53:07.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":1,"date":"2020-07-22 20:42:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-23 19:41:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-23 12:35:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-22 20:42:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 08:51:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-18 01:35:39.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-20 14:37:09.000000000","updated_by":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-07-22 07:06:07.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2020-07-22 07:09:25.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-07-22 11:31:09.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 11:49:11.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2020-07-22 19:52:28.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-07-22 20:34:51.000000000","updated_by":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"reviewer":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"state":"REVIEWER"},{"updated":"2020-07-22 20:42:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-22 20:42:42.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"},{"updated":"2020-07-22 20:50:53.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 21:12:53.000000000","updated_by":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-22 21:47:38.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 02:53:07.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"REVIEWER"},{"updated":"2020-07-23 02:55:40.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2020-07-23 03:38:37.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 12:35:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-23 18:27:43.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 19:41:57.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-07-23 20:01:35.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 21:55:19.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 22:02:20.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 22:05:08.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-07-23 22:18:46.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 22:28:21.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-07-23 22:42:39.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-07-23 22:45:00.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 22:46:15.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 00:20:27.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 00:21:03.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 00:40:15.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 01:28:08.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 01:29:25.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-07-24 01:29:49.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-07-24 01:33:56.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 01:47:11.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-07-24 02:03:20.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-07-24 02:05:22.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-07-24 02:10:52.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-07-24 02:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2619e4e50634e5d1bfbeb85dec7fe4047f413d9","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-15 11:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d307798e76e759a1ed54bd625fd4d2b4218d14e","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-15 11:30:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/2fc6eae95c17d3978e7f9a8645ea9af8b374229b/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1635ec2ddfb2d1235f2ceee063207215ecee7eb3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-15 11:57:38.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/1/check-not-vote/ext-cinder-zuul/c4ce5c5 : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5028ad9ed233a3932cf51be09dbacb04e49d886e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-15 12:02:48.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ad84f91 : FAILURE in 20m 48s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6aa73eb : FAILURE in 19m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"9c5f3c5b036381dff8b0882da6ced899daa4b136","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-15 12:09:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/1/check/dsvm-tempest-cisco-zonemanager-job/311cfb9/ : FAILURE in 32m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9781d7b29140c0ddef305aaaff502192ccb77137","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-15 12:15:09.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"52a6f26f5547fa1aca40a5b2abe750242729a18a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-15 12:35:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/884f614/884f614.html : FAILURE in 1h 08m 57s","accounts_in_message":[],"_revision_number":1},{"id":"eefe29c73c8992fa05d3db71f65a6b719712f955","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-15 12:35:33.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/1/iscsi/ : SUCCESS in 1h 11m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/1/smb/ : FAILURE in 56m 35s","accounts_in_message":[],"_revision_number":1},{"id":"dda37f4fc136f47f9bc0cbf0ac354bee0b1960d2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-15 13:21:47.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fd792cce06888e6ef82be9c8193ec8f1bbd90a65","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 13:31:09.000000000","message":"Topic changed from bp/powerstore-cinder-driver to victoria-drivers","accounts_in_message":[],"_revision_number":1},{"id":"1dfc6274ffb5ee751f0bbd29ff4841e77efa0b83","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-15 13:32:54.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-1/ns5_nfs : SUCCESS in 2h 8m 35s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-1/ns5_iscsi : SUCCESS in 1h 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0de74892d9d74a06df3a9b53dd4ea9ebe7ff674d","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-15 13:34:00.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"54d34bef41101b3dd015c97ee3a1614f1a83966f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 13:38:30.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nLooks pretty good so far. One issue pointed out. Still needs release note, driver docs, and driver added to the support-matrix.","accounts_in_message":[],"_revision_number":1},{"id":"e24d442f1b22decf4a9c9ac8ae7187f19b42de77","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-15 14:32:03.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/1/check/DellEMC_XtremIO_FC/d0b177f/DellEMC_XtremIO_FC/399 : SUCCESS in 1h 29m 36s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/1/check/DellEMC_XtremIO_iSCSI/8f6fe4d/DellEMC_XtremIO_iSCSI/371 : SUCCESS in 1h 30m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"922d2431af0bbc165fbf05c0eec87d163c67e274","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-15 14:35:18.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"47d78cc4ccf0b7274419d04b118db26a999e9027","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-15 15:49:51.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/58/741158/1/check/DellEMC_Unity_FC/504414c/DellEMC_Unity_FC/298 : FAILURE in 2h 50m 10s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/58/741158/1/check/DellEMC_Unity_iSCSI/da64862/DellEMC_Unity_iSCSI/280 : SUCCESS in 1h 32m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"384fa9a4eb55ea707dd81ff2ade479ad6d001543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:12:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e9ba732687f4b4596d41d977d560c7b : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75270b47c04c46edb8fe06cc9f7c5291 : FAILURE in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e854799136124c6dbb3090a91d763b5d : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7c66198083c4e90b1d368627d1c5e3e : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9fb33b0b6e144b5b91214cad44de97e : SUCCESS in 16m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/e96f702e8c3c4104a42c7e8e143c834f : SUCCESS in 1h 07m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb185a18b9424ff9940ec789ddb95ea9 : SUCCESS in 1h 14m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/133135c0f786443e959769cb38f575f4 : SUCCESS in 14m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb8c220baa494b209074120ae13d43ee : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23eccf05262048d79cf3116ca426ebc2 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/640b7e66249b42bbab2e1e0a1ec8ada5 : SUCCESS in 8m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2c3ffe06002b4d5bb9cc8832d892f91b : SUCCESS in 48m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/acd852cf192e4721933e20ea30f101b5 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/077e58b266f44a69ac47d8fcd0d2d062 : FAILURE in 1h 07m 40s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f0118f2de5d421db698b5e88c6b7afc : FAILURE in 1h 59m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/db197cc062314981aa0dff3e9c64a0d6 : SUCCESS in 1h 18m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2da4033e76d145cb9b2496b96ee119a7 : SUCCESS in 1h 20m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b4999499efc549cd9aba97ad7ec69d31 : SUCCESS in 1h 08m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c5520a6e1dd4ef7820d635e0469d22b : SUCCESS in 1h 59m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f93ba6bfe6a4ebfb52fd6be57f4dcb8 : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"eb0bd8ed7c30fafb3d740ab3dbcb04fd83c2437f","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-15 17:18:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c58cf8dd73e14fc1d3732b5e61587eff5da2b125","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-15 17:23:15.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)\n\nHi, Sean.\nThanks for review.\n- Support-matrix - done\n- release-notes - done\nStill working on docs and unit-tests.","accounts_in_message":[],"_revision_number":2},{"id":"c46fd21345fa7cce89bf65e37e3b8f1f99728385","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 17:31:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df72f6413591d36b093572e564a24f10f45dccb3","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-15 17:42:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2b701c76ccc65ee9d49069158e01cc884d2c3830","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-15 18:10:16.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f2e36ed4ef43887b210473fdfaed159615c48a1c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-15 18:34:10.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"27d43761ec35f1473cf61a9e424ad3964796a733","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-15 18:45:24.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/3/iscsi/ : SUCCESS in 1h 02m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/3/smb/ : FAILURE in 54m 19s","accounts_in_message":[],"_revision_number":3},{"id":"7b079e4e585eab83adbddef6e5400503132d76e8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-15 18:50:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/b377cf7/b377cf7.html : SUCCESS in 1h 07m 06s","accounts_in_message":[],"_revision_number":3},{"id":"5224bc255c00b65cd8430fd742fe9d1109556ff3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-15 18:57:43.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bbea97294b4548cc0aa9784aa7f2d21dede66f65","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-15 19:22:13.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-2/ns5_nfs : SUCCESS in 2h 3m 4s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-2/ns5_iscsi : SUCCESS in 1h 50m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9f45fe1984e1ac5599a13460b532db3dc704b821","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-15 19:44:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/3/check/dsvm-tempest-cisco-zonemanager-job/e2995f0/ : FAILURE in 33m 41s","accounts_in_message":[],"_revision_number":3},{"id":"90e4a7ddd9e2974585a2f4197348acfe9b12582c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-15 19:48:57.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-3/ns5_nfs : SUCCESS in 2h 2m 51s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-3/ns5_iscsi : SUCCESS in 2h 6m 8s","accounts_in_message":[],"_revision_number":3},{"id":"efac2be9455fb4d907e0874661f70c4516c31622","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-15 20:09:39.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2c1bd08df5606bbcb11cd2f93cb40722576cd7ef","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-15 20:16:29.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"1381a9996d2167cb92fd20dcad011829835e52c0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-15 20:20:13.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ffdbc76 : FAILURE in 19m 58s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/5d8fac0 : FAILURE in 28m 38s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"816d39c2ff05bcb17412b6d0878a4473b8054f16","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-15 20:54:50.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"1c81ad1c9c6da5e40e5f9acae6afc22b384eff88","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-15 22:12:11.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/3/check-not-vote/ext-cinder-zuul/8b69f77 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":3},{"id":"a4ce3393194c805a983c2d83c3a080f7db930aeb","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-15 22:24:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/38725da45a8b7323b6df6f9d625d9e8e77ddd452/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"326b8d19beb3cf5d30198854a660706c8451b94e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 23:02:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3c0d09c6a4643b7919bdf8c34ad8a90 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6974abbeef104a9d9c9580ad8e633b36 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b4e38b536894637b5f324eeaf677b40 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/612e56fa368b4a618cb12773eeeee184 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d98e1ce8ec364043a73ba0102bf8afb8 : SUCCESS in 15m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/7010e99ec6c84db68ed5f4c810d50e67 : SUCCESS in 1h 14m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/27ed5296a1ca4bc6ad7ab14b706dc9ab : SUCCESS in 1h 29m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6554f72db81c437abee523ed4e647fed : SUCCESS in 7m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/68d2e5e066004a82be4bc4339d99880d : SUCCESS in 12m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4662c76616d6429facf2da92a95fbffe : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3dfad29f245c4f6eaf617bf898ee35e3 : SUCCESS in 7m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0a719b48bfc44dd493dff06003a26f09 : SUCCESS in 10m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c62a5035943844279a84825812a9897c : SUCCESS in 31m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a4812289cd9a499493b11e3442b0f6fc : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dcae131c4d404b7d91946dec21d58c33 : FAILURE in 1h 07m 00s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a7da1b03b4e842ae8944c92beff3260f : FAILURE in 1h 37m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/35ea8a1119dc473abaeb0fe48586706b : SUCCESS in 1h 14m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7a2a1e763e934020a66da15e2b7bda86 : SUCCESS in 1h 18m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2a5266bda1244767b1c18e757f488e93 : SUCCESS in 1h 12m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d614c95aec0041c09ce27b92fd74b22d : SUCCESS in 2h 07m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b60b0b0aa779417e8f68f0f1cccc4120 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":3},{"id":"75f79576385e6c79f43c494b71edb3f63e45407a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-16 00:03:12.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/3/check/DellEMC_XtremIO_FC/8eb3da7/DellEMC_XtremIO_FC/405 : SUCCESS in 1h 35m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/3/check/DellEMC_XtremIO_iSCSI/05b8b15/DellEMC_XtremIO_iSCSI/377 : SUCCESS in 1h 33m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"c5d704020bb4d8f717afc367fd1f0c0553c778b1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-16 02:34:24.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b9aa1fa950d0b3ee8bf6a7983591dc7e0e6b6f71","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-16 03:29:11.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/2ff207a59ced8ed927c4e74bb153287d33ffac0b/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"7b13962bc5aef359b7954210125059cfc54f3ecc","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 04:17:31.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"9f8fd2abf5b3466951427d2cb492674ea703d7b8","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 04:19:32.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"e5bc0e25406c2e8f15b90e6ec14b48e959d33f55","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 04:37:34.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"b36f59efc53a29b58c9b65badb58ffa1bca80b17","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 04:50:51.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"896c22cf0d63d1bdeb50f68e513627ea3f6ea0e1","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 05:14:43.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"01fc224ee320cec02679c32d2f5734943f2c8262","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 05:30:44.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"891f674171b80a7c7a7e90ea2f010ef17efb5e61","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 06:20:32.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"b960ad43c6e1d9dd0573fce7abaadac8fa4c1e00","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 06:26:38.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"31a600372deb664fad9f8ba70a08d63e2b6a8a41","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-16 06:26:54.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/3/check/DellEMC_PowerStore_iSCSI/7445519/DellEMC_PowerStore_iSCSI/None : NOT_REGISTERED\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/3/check/DellEMC_PowerStore_FC/494d5c4/DellEMC_PowerStore_FC/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"6ebfc14cfbbd2090e95116cef00216f76c4faa1f","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 06:36:03.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"5c4bc089fa8c1acb149bc4cac09c6f32823007e2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-16 06:36:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/3/check/DellEMC_PowerStore_iSCSI/d020c96/DellEMC_PowerStore_iSCSI/1 : FAILURE in 5s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/3/check/DellEMC_PowerStore_FC/54884f5/DellEMC_PowerStore_FC/1 : FAILURE in 5s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"282738d57829f8aa32177700d25c73d82cf030a1","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 06:42:14.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"78ce265ec1dba6c2b6cd4701b28e21f567707a65","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-16 07:25:41.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"380a1237bafe2499631c1ee7a4c292c64a481290","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 07:42:57.000000000","message":"Patch Set 3:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":3},{"id":"39147135635381a9f0ed0cc03098692d10e96ed1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-16 08:51:27.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/58/741158/3/check/DellEMC_Unity_FC/4d460b6/DellEMC_Unity_FC/304 : FAILURE in 2h 50m 31s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/58/741158/3/check/DellEMC_Unity_iSCSI/11aad7e/DellEMC_Unity_iSCSI/286 : FAILURE in 1h 32m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ddb9ab34c9d59efd856afc113fce4dc37f6acdeb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-16 09:35:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/3/check/DellEMC_PowerStore_iSCSI/cb8884f/DellEMC_PowerStore_iSCSI/3 : FAILURE in 1h 51m 13s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/3/check/DellEMC_PowerStore_FC/1d4a1a2/DellEMC_PowerStore_FC/3 : FAILURE in 1h 52m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"1f7ea578ed1ac81ea94ebddfaf70508077fecf81","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-16 10:15:25.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"904136fb7b73aa7e9ad7a8ed3935c9f1ac3ea804","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-16 12:31:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5eedbbe32f02e93bfa87b6b72fb7286fbfcf2ed3","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-16 12:33:50.000000000","message":"Patch Set 4:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":4},{"id":"edbca841970cc7fe3adc4c541d41e60fba729b2e","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-16 12:38:24.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/06aafab43acf42cc2bec8d95c85e1fafe94c174a/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e32d210d6a40c1814791e78788894dce0e0a41fd","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-16 12:38:57.000000000","message":"Patch Set 4:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":4},{"id":"dcca6203cb79686f09150493a9fa168097f75bf6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-16 13:07:52.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/4/check-not-vote/ext-cinder-zuul/3d71f1b : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":4},{"id":"a8be41664a07ccb462758fe962b179cd1bc0dda7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-16 13:21:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/741158/3/20200716194727/NEC-ISCSI : SUCCESS in 2h 26m 05s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/741158/3/20200716194727/NEC-FC : SUCCESS in 2h 33m 44s","accounts_in_message":[],"_revision_number":3},{"id":"9623ee24f5d550f601c535f9534540a0e4942e05","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-16 13:49:17.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1689 : SUCCESS\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1897 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1725 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"fce79cb726666efdda787fde63e538f291ab6cdc","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-16 13:56:28.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"02d77cb02c5021601585e82c689a7d6cb772217f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-16 13:58:36.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/4/iscsi/ : FAILURE in 1h 13m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/4/smb/ : FAILURE in 53m 57s","accounts_in_message":[],"_revision_number":4},{"id":"e14169b18dee07f75dcbb3163e1bc38c16a787e4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-16 14:21:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/758e8f8/758e8f8.html : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":4},{"id":"1b76ab0ae59fa85c11e631b447b2db6211a32468","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-16 14:42:40.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-4/ns5_nfs : SUCCESS in 2h 3m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-4/ns5_iscsi : SUCCESS in 2h 10m 26s","accounts_in_message":[],"_revision_number":4},{"id":"551e14bd3933920e24d19f6c5f25d2dfe1e2242b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:54:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/286317033c2b403988780a5cb5343e68 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b46d2fa465e4e5fbb07e55610bac589 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba6b5ff44b1c454f90e5526ad1d45e89 : SUCCESS in 11m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dda7996b73b44868b639b2489d5d7f39 : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd27e9ca72549dcb4921721fb845df6 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/b05e4cd3620f46aeb8e1e2d6b3fe8e25 : SUCCESS in 1h 10m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a11210fbcf594df199b8d71536f3aa1c : SUCCESS in 1h 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed915bcf07fc4503a220ddc7ef028f05 : SUCCESS in 8m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8306ec30aefe409d9b5fd9960212b4bd : SUCCESS in 13m 22s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86a4d44be3c54470ae0b66892f67ca95 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa9445087d2847c596b0a77cc5e1bb68 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b134e308b82e4cb8ad64eec9af091d1d : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ddc554c8c964479caa403f9e263d70d0 : SUCCESS in 43m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be2f6d6cb07f4458af139912fb924621 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d132703c0c2644b385a24967e72a8184 : SUCCESS in 1h 12m 49s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ca22905313cb4a839c92e932976a2fee : SUCCESS in 1h 32m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/07c57a722be441da873b7d9a3dc89f14 : SUCCESS in 1h 14m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b57e80ca1eb0411cb9a7cbefd9a3a683 : SUCCESS in 1h 21m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f9ba68bff7bc49058da292c513fecdd0 : SUCCESS in 1h 26m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b77b114d801e4e808a9550819d362cfe : SUCCESS in 2h 15m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1246c275acb4956a22d05a27e1d0cb4 : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":4},{"id":"440dc98090e1960572e70a71afa3c7c16ad607d2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-16 15:02:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f57901e : FAILURE in 20m 07s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/81765aa : FAILURE in 21m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"3a14cff9e74207fc5226eaec8e5633d80ba940b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-16 15:22:32.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-741158-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"008b5337f7db0648db5b094368674ad110ec7123","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-16 15:29:29.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/4/check/DellEMC_PowerStore_iSCSI/e983b21/DellEMC_PowerStore_iSCSI/7 : FAILURE in 2h 50m 19s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/4/check/DellEMC_PowerStore_FC/a09a2c3/DellEMC_PowerStore_FC/7 : FAILURE in 1h 21m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"af90ed212f08c5e1eff38f9659b81fce3f4ac64c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-16 15:50:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/4/check/dsvm-tempest-cisco-zonemanager-job/d2f8490/ : FAILURE in 32m 36s","accounts_in_message":[],"_revision_number":4},{"id":"89dd71edde3451ba1af96635b3d55aa5351d6d6a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-16 15:55:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/741158/4/upstream-solidfire-check/cinder-solidfire-iSCSI/f4f1c25/ : SUCCESS in 59m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"066d84bcc69cf528496365c68d6a62aa957e6591","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-16 16:06:36.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e067b2fbcda12fbf43b72115fec8e0980f648567","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-16 17:38:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/58/741158/4/check/synology-dsm-cinder-iscsi/4f257e3 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":4},{"id":"e4ce2e050e1ff0405a45ceb1d79de9f3fd9b1203","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-16 19:11:42.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"0467966022c62f24e3e70a35d7091b9d7bc3b9ba","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-16 20:32:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c2844b45056d5e32fceec114f599b142e568a17","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-16 20:37:33.000000000","message":"Topic changed from bp/powerstore-cinder-driver to victoria-drivers","accounts_in_message":[],"_revision_number":5},{"id":"dc7420018aecb512a5a434bf30ecac55cf52a3ac","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-16 20:59:27.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/5/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3b4c8bf : FAILURE in 26m 33s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/5/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/8bff9a1 : FAILURE in 20m 33s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"c95925ed3710ca3e16709797aa50f3923904821f","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-16 21:02:41.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1903 : FAILURE\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1697 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1731 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"5d1c5d994c6d18b3effe4d012461e172ea654c62","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-16 21:23:38.000000000","message":"Patch Set 5:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"bd96e6b6b6e80c3f26d94e1dcac2db3a48e9e1c4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-16 21:24:36.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/5/iscsi/ : FAILURE in 40m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/5/smb/ : FAILURE in 51m 36s","accounts_in_message":[],"_revision_number":5},{"id":"1b78d5cc383c4d847822a39c411bbdaa37217780","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-16 21:34:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/5/check/dsvm-tempest-cisco-zonemanager-job/266b1af/ : FAILURE in 32m 40s","accounts_in_message":[],"_revision_number":5},{"id":"264004378c63c1c52d2f8f3b1c8c471ebe57ffcf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-16 21:42:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d2bc891/d2bc891.html : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":5},{"id":"74bb56dd7734e192ddaa9d3a688d3ea4ebc172bd","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-16 22:31:06.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/fba5a0b10e4910c5af8391d0d87208254d5d3013/index.html : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"b8dd76b403373ab0b41b5a0485fd3ecec521bb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 22:37:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/341dd1dfdf4d4da4810934ce63c7daa9 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88cfed03ba4b4da9ae98934cc1b18a49 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21335ecc9bd648f4870effc08d93165d : SUCCESS in 10m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d855f02d407e49bbbd7be5819bcad490 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b149da9d27428d892e47e641d98141 : SUCCESS in 16m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/d12904cd62514b3083c027f7424e92e9 : SUCCESS in 57m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01bb769da52e409e819f924c731fc72f : SUCCESS in 1h 03m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e6b73a185424ad18c05cc7dfc75823b : SUCCESS in 8m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29633bba6c534257b2c83157ee6b4012 : SUCCESS in 13m 21s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3ffe7c1b327447a3addfc1acb9c16bfc : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ed355ab207c408e9b2baeaf8d0aee7e : SUCCESS in 9m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/93ec60ff82e747f0b5738b6e02511128 : SUCCESS in 9m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5eae5c4fbd04d518d312e65e1508f8f : SUCCESS in 42m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6e8ad4da02ca4ebb84bdd17431904ee8 : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a0c948131be344d9a5139c490ff3085f : SUCCESS in 1h 23m 51s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee52804763744fe284b90c0f102f5b31 : SUCCESS in 1h 23m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/10789106bff4494180a306e0c9b0c5a5 : SUCCESS in 1h 08m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b9b20ef4e2ff4e919fde075fbbc0a0e2 : SUCCESS in 1h 17m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4a3584632f2d4a6c9fc0b0de3db45095 : SUCCESS in 1h 07m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed338245e3c545b08920a9347ac5e93b : SUCCESS in 1h 58m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/753343f4ed2945369b31944809dbb939 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":5},{"id":"07aedd3e8a0768c4c22687de2489894720d6e6c1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-16 22:41:25.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-5/ns5_nfs : SUCCESS in 2h 5m 55s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-5/ns5_iscsi : SUCCESS in 2h 8m 31s","accounts_in_message":[],"_revision_number":5},{"id":"c957c37e684ea88f79829555f63eac8f1a7f92d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-16 22:43:46.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/5/check-not-vote/ext-cinder-zuul/4711be5 : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":5},{"id":"9af2cbbe15aaa106415f7f153fda2a82887f4e23","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-16 22:44:33.000000000","message":"Patch Set 5:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-741158-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"3bae76949c519f3a67e4c548030bcd2de46d0082","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-16 23:38:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/58/741158/5/check/synology-dsm-cinder-iscsi/64c0eb9 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":5},{"id":"9ca30598b21283cd0acecdee1a1477c48854c7e4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-17 00:17:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/741158/5/upstream-solidfire-check/cinder-solidfire-iSCSI/fb96271/ : SUCCESS in 1h 38m 07s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"1068c6e42fab2a8cf9981aef915e4f66b37408c6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-17 00:36:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/58/741158/5/upstream-check/cinder-cDOT-iSCSI/f85e510/ : SUCCESS in 34m 59s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/58/741158/5/upstream-check/cinder-cDOT-iSCSI-IPv6/117fed4/ : SUCCESS in 1h 51m 10s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/58/741158/5/upstream-check/cinder-cDOT-NFS/324dc1d/ : SUCCESS in 36m 29s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/58/741158/5/upstream-check/cinder-cDOT-NFS-IPv6/a508dcc/ : SUCCESS in 1h 57m 20s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/58/741158/5/upstream-check/cinder-cDOT-FCP/b1e14e9/ : SUCCESS in 51m 50s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"a3f58a8b9f9770c6addd57d38ae5ef7d1b6e186f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-17 01:48:12.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"861334ae9e9302751d296e97930e083c0b50588d","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-17 02:05:11.000000000","message":"Patch Set 5:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":5},{"id":"f24c900c1b0de27d63ecf2bf03298ac5e4da2702","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-17 02:08:10.000000000","message":"Patch Set 5:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":5},{"id":"dfe31f29863f33483480b8be44a55049bdd5e76e","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-17 02:09:37.000000000","message":"Patch Set 5:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":5},{"id":"ba3949751d702b9e6d220054cafaefb85b8ba813","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-17 02:34:53.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"7d973940d5fa25689527cdca684d471bc74d531d","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-17 06:56:33.000000000","message":"Patch Set 5:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":5},{"id":"cdb484297dcddb5252814de9ee67077a1720c112","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-17 10:58:55.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/741158-4/Seagate-iSCSI : SUCCESS in 43:00.62\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/741158-4/Seagate-FC : SUCCESS in 41:38.89","accounts_in_message":[],"_revision_number":4},{"id":"8bd01d6d0c24cf1885067831e871f30b54fdced3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-17 11:59:47.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/5/check/DellEMC_XtremIO_FC/35e51a5/DellEMC_XtremIO_FC/427 : SUCCESS in 1h 34m 32s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/5/check/DellEMC_XtremIO_iSCSI/cbf5d7c/DellEMC_XtremIO_iSCSI/399 : SUCCESS in 1h 36m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"522ae510bbd0964c064e19ac11a45260e561fe08","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-17 13:14:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b382ec04ed6598105069d91d742c873b5965da0e","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-17 13:15:51.000000000","message":"Patch Set 6: Workflow-1\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":6},{"id":"c1758a0156d53c2b82fbfd84861d673a23acb170","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-17 13:20:21.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/d8831866936c00f37f54252e858bc9a04c7e0e15/index.html : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"225204d7f7c4ffffeb7485e2b8b41433bd7c2bb3","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-07-17 13:41:27.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1909 : SUCCESS\n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1704 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1737 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"a8c1cf09eb716d9f1f9f7cf517633ba50068aee4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-17 13:52:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/6/check/dsvm-tempest-cisco-zonemanager-job/e934676/ : FAILURE in 32m 18s","accounts_in_message":[],"_revision_number":6},{"id":"0ba421e5414f297373e4ff506e7bf67d76544e55","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-17 13:54:58.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c63b461d4510384266f64d43803cc0e48a17570b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-17 14:06:15.000000000","message":"Patch Set 6:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"af9b6314fb19743f0f63eaafabad702dbce6aec2","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-17 14:15:00.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/741158-4/HPMSA-iSCSI : FAILURE in 51:20.60\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/741158-4/HPMSA-FC : SUCCESS in 47:46.06","accounts_in_message":[],"_revision_number":4},{"id":"bf428764a98a85f6cdc6d477d75c940943aec716","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-17 14:23:11.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/6/iscsi/ : FAILURE in 1h 06m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/6/smb/ : FAILURE in 1h 07m 58s","accounts_in_message":[],"_revision_number":6},{"id":"53c067f106a49865f8ce15124749fe124ee1ffac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-17 14:23:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/35a43ec/35a43ec.html : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":6},{"id":"bf9a5dac2940ee1c4910aa200a805e8e9425408b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-17 15:08:48.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"6ee996f984ab009e3c0608f5ced997b8548d2df7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-17 15:13:25.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/6/check-not-vote/ext-cinder-zuul/e965da3 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":6},{"id":"550f4ca44bc8dbc4a2f73b36d00712f5eab0bcf2","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-17 15:13:38.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/741158-4/Lenovo-iSCSI : SUCCESS in 53:43.64\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/741158-4/Lenovo-FC : SUCCESS in 48:00.67","accounts_in_message":[],"_revision_number":4},{"id":"fa05e973801358c927c6b7a07cb1dcf6383b4229","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-17 15:23:51.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-6/ns5_nfs : SUCCESS in 2h 6m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-6/ns5_iscsi : SUCCESS in 2h 8m 48s","accounts_in_message":[],"_revision_number":6},{"id":"612b10fcb429666844e0046188c7c34ff7cd1d73","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-17 16:36:23.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"214c36fefe247e588a689837d8aa3dac57e693d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:45:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/595fa5b0839f486692b5128def69759d : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e73b3754f324e7e9560e711bbc7cc8b : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba0be70599b140bdbd5236e5ed10d2a3 : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dbc08d5f4244e76aacf9949da8f9591 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22f72a91fa8482791241c47e69a7fda : SUCCESS in 14m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/fd71e54b87a546bc9a8f2c47cb409114 : SUCCESS in 58m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f098e0216abf4752baaa75072879d197 : SUCCESS in 1h 00m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43e735b227a14eeca893c47c576e1779 : SUCCESS in 8m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9e6e4a597d754b6199e56da17f425bde : SUCCESS in 14m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1600e80da8e3498ea0781a2df6ebabac : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d26d2b34f5204dcba21d898524d28d13 : SUCCESS in 6m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e91ef0d0e08149d2856623c44ae9ee73 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f5922add303947a893e29f1bde391e62 : SUCCESS in 40m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d99fcccce3a847a79714bdaab6a447a8 : SUCCESS in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8c2bc568a6e54448b0f99317ccc2ad35 : SUCCESS in 1h 04m 31s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1878c39af8248279e5dfe7afdabe1af : SUCCESS in 1h 02m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c8ed42d2bcc4162880d818b4405f4fe : SUCCESS in 1h 15m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1707e428369e4383b378bb9c08451d30 : SUCCESS in 1h 22m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2e691743764b4504814a4de39727b617 : SUCCESS in 1h 14m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bfc24314f54a468e9eff18c969ed92ba : SUCCESS in 2h 06m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdd53fd903aa472e932a21a83ad0002a : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":6},{"id":"498d17f387b63c932596b908aee65b910a61a6af","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-17 16:52:33.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e69356e : FAILURE in 30m 50s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/c9d3d9e : FAILURE in 26m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"711a4a7cd9d5e10d91dd685643930f0958e99665","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-17 17:10:43.000000000","message":"Patch Set 6:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-741158-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"b3b5ed95d43545384b858298db798212700f7ec2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-17 17:10:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/58/741158/6/check/DellEMC_VNX_iSCSI/26e1763/DellEMC_VNX_iSCSI/155 : FAILURE in 25m 30s\n- DellEMC_VNX_FC http://publiclogs.emc.com/58/741158/6/check/DellEMC_VNX_FC/f6da90b/DellEMC_VNX_FC/148 : FAILURE in 24m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"9ca23b244001d3d102152b70ada35ae5d495920b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-07-17 17:12:21.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/58/741158/6/check/DellEMC_SC_FC/8f2974a/DellEMC_SC_FC/194 : FAILURE in 26m 56s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/58/741158/6/check/DellEMC_SC_iSCSI/53fe387/DellEMC_SC_iSCSI/166 : FAILURE in 25m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"7305bf95260621927082d5bbf856720e83959a87","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-17 17:15:17.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/58/741158/6/check/synology-dsm-cinder-iscsi/c288ac5 : FAILURE in 29m 37s","accounts_in_message":[],"_revision_number":6},{"id":"b6e8f4eb101d9f749502006ba2ff3c7c19a54b9d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-17 17:20:29.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/58/741158/6/check/DellEMC_PowerFlex/fa0a9aa/DellEMC_PowerFlex/274 : FAILURE in 24m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"0dda0eb396ae7f35c5f15e311535b5dfa6c92ec2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-17 17:31:57.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/741158/6/upstream-solidfire-check/cinder-solidfire-iSCSI/daecf9f/ : FAILURE in 45m 39s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"cd2ac712fb520a8b8e4231bb12f05611143379a1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-17 17:46:53.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/6/check/DellEMC_XtremIO_FC/9776605/DellEMC_XtremIO_FC/434 : FAILURE in 32m 34s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/6/check/DellEMC_XtremIO_iSCSI/9f5974e/DellEMC_XtremIO_iSCSI/406 : FAILURE in 26m 37s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"98a17b9c3c7d14c740aeab93c64cd6965307cced","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-17 17:53:19.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/6/check/nimble-iscsi-driver-dsvm-volume/e7e84ac : FAILURE in 1h 11m 30s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/6/check/nimble-fc-driver-dsvm-volume/5c10369 : SUCCESS in 1h 56m 55s","accounts_in_message":[],"_revision_number":6},{"id":"f4f01ad97cf069768d7c36cf1b82b81b2978b912","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-17 20:54:27.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"7eae911527b1b8c3ece0e791ba64e55f47f9c59b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-07-18 01:35:39.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/58/741158/6/check/DellEMC_PowerMAX_FC/781d599/DellEMC_PowerMAX_FC/311 : FAILURE in 34m 55s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/58/741158/6/check/DellEMC_PowerMAX_iSCSI/5958e90/DellEMC_PowerMAX_iSCSI/249 : FAILURE in 26m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"0b2b5b93bc303baa9ba0378f916dd566ebe267f1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-18 02:30:40.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"0653b540fe33973c70e400b63ee80a3f2840df0f","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-19 09:51:59.000000000","message":"Patch Set 6:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":6},{"id":"4065907f87a9752413dd8bc57423092494ce2f23","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-19 10:26:03.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/6/check/DellEMC_PowerStore_iSCSI/6419d87/DellEMC_PowerStore_iSCSI/15 : ABORTED in 33m 47s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/6/check/DellEMC_PowerStore_FC/e60b842/DellEMC_PowerStore_FC/16 : FAILURE in 12m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"efa0a6d222f6e4ce654672345083eda3ff3c89ba","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-19 10:26:31.000000000","message":"Patch Set 6:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":6},{"id":"0b3893f65e42d5c52c0ee335fe7cfd44bdfcb230","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-19 11:02:24.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"5d2a715475b3bc10729336cf618e516c6c839382","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-19 13:14:28.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/6/check/DellEMC_PowerStore_iSCSI/947ba9a/DellEMC_PowerStore_iSCSI/16 : FAILURE in 2h 47m 44s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/6/check/DellEMC_PowerStore_FC/50e96ef/DellEMC_PowerStore_FC/17 : SUCCESS in 1h 18m 59s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"36d076e09e7f988fcd5bed5cd9a91c08473b8ad9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-19 13:48:08.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"933c01cb476e2834b8bc4e6edff71ce60ba0185b","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-20 09:43:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2b54e30a6cdc901d82dfc923d60303a3d6ada18c","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-20 09:48:04.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/0f43707ce33a9a03eefb073d51776106f82993f7/index.html : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"1325d8c47ab7004449240814058d1f2a3a4ee2de","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-20 10:04:10.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/a3f44ef : FAILURE in 20m 45s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/bc8ad9d : FAILURE in 20m 42s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"393beba71bf6990770e3a226d9349a20a1c5b0b1","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-20 10:05:45.000000000","message":"Patch Set 7:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":7},{"id":"10a79418f006e6e1cb684b7c520b063ef5f9f63b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 10:15:39.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"bde65a5bb2c0153a2a5a4c963b114a4b4b7d365e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-20 10:19:24.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/7/check-not-vote/ext-cinder-zuul/a6d3d1e : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":7},{"id":"fbe0732c0d43253fd2dffa1f28ae1790f8da8eb8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-20 10:20:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/7/check/dsvm-tempest-cisco-zonemanager-job/7e46bb4/ : FAILURE in 32m 24s","accounts_in_message":[],"_revision_number":7},{"id":"33b604eb4fdc9e8fa603daabd686ee69538bb22c","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-20 10:34:17.000000000","message":"Patch Set 7:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"b27cf2cfa0ba53353753465d1bd8e1cc4b14003a","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-20 10:42:00.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1707 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1915 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1741 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"561e716f4a62262d6325099535991afc8833ab91","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-20 10:48:20.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/7/iscsi/ : FAILURE in 1h 04m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/7/smb/ : FAILURE in 52m 34s","accounts_in_message":[],"_revision_number":7},{"id":"46d1ff4232a11ab20cc34b8f823588024266f6e0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-20 10:53:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e768424/e768424.html : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":7},{"id":"14e26be35c1e8a9a5a3d8cc925ea149ab3301799","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-20 10:54:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/7/check/nimble-iscsi-driver-dsvm-volume/a993750 : FAILURE in 1h 11m 35s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/7/check/nimble-fc-driver-dsvm-volume/f81bcea : FAILURE in 1h 02m 18s","accounts_in_message":[],"_revision_number":7},{"id":"2001b1ddb31db0b6efe4b2213bf371a076c3208c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-20 11:17:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/7/check/DellEMC_PowerStore_iSCSI/4b845dd/DellEMC_PowerStore_iSCSI/17 : FAILURE in 1h 02m 41s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/7/check/DellEMC_PowerStore_FC/fa770bb/DellEMC_PowerStore_FC/18 : SUCCESS in 1h 11m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"8093947c55ae573d6f9a3b7e8c8cc79be714cabf","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-20 11:18:54.000000000","message":"Patch Set 7:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":7},{"id":"b5b4fc93411b21a01df11822ad9e36504c5c3336","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-20 11:19:36.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/7/check/DellEMC_XtremIO_FC/463489f/DellEMC_XtremIO_FC/439 : SUCCESS in 1h 32m 35s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/7/check/DellEMC_XtremIO_iSCSI/0d6f23a/DellEMC_XtremIO_iSCSI/411 : SUCCESS in 1h 36m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"5408ab5e1f43ee6dd9aadcde392faf4bb9b1c36a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 11:24:06.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"262fb97112e28fced50a779df1d52180d38f83b4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-20 11:49:04.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-7/ns5_nfs : SUCCESS in 2h 4m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-7/ns5_iscsi : SUCCESS in 2h 5m 38s","accounts_in_message":[],"_revision_number":7},{"id":"e154d12f9bfa9728678c03e7270f3888eb773ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:04:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dcd309139534fc5b8ab282ab82b53df : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/184a5eaeb84340c8b702701ca14806f5 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4e5e724b3e543d9b1f097ca5d4bf32e : SUCCESS in 9m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70da71e5447142feaf8cd805ac45bdc2 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5ce9986ea6b4de3895b8d313c6b147b : SUCCESS in 14m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/2e223cbb64584830a76d66ab5cde6e0b : SUCCESS in 53m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a79659c8073145be8bba6db396319796 : SUCCESS in 1h 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fde301f3900d481ea8041fc87ade2972 : SUCCESS in 9m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/07ab0bcc87564c67abb66160dfa5c489 : SUCCESS in 13m 07s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/560e3dfb3c934d87a31c7542bbcddcc4 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fff8f4c6e79c4b5a8989326576e9acba : SUCCESS in 8m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a250309ddfcc4de8a5b0c443f8c1fa75 : SUCCESS in 7m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f1d2f2d98e4f4b29b21f316f49510fe6 : SUCCESS in 34m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7defd5a8d15f46a0a42499b75ab1ac74 : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c96633773ac477fb165ba9bcabce17d : SUCCESS in 53m 31s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f61166fafdfc4b99940ad90e7015c8a8 : FAILURE in 1h 02m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/196b7709444a47d09e270a6aec0497c7 : SUCCESS in 1h 19m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/32335fccb3434a31b2d8327622cee627 : SUCCESS in 1h 22m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/028d17c499a8499b9c350089e707686e : SUCCESS in 1h 06m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/599d37dd9c264ba6b135d4fb6a4fcd53 : SUCCESS in 2h 12m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05b7c8b9ab4c44a78523a82d2e9a5c77 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":7},{"id":"a5a25b8d682809f612730def79a5a972b4cae12b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-20 12:27:05.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/7/check/DellEMC_PowerStore_iSCSI/9cd3650/DellEMC_PowerStore_iSCSI/18 : SUCCESS in 58m 57s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/7/check/DellEMC_PowerStore_FC/ef190e3/DellEMC_PowerStore_FC/19 : FAILURE in 1h 07m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"ac78e28ddb8f2ae6a4173ed8c52c286c386995c3","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-20 12:39:59.000000000","message":"Patch Set 7:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":7},{"id":"9fa8129c274d72e86af25e71385a23b11c898f7e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 13:01:26.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"6cbec098e06c3d588cce0d17538ab31f7c8545a1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-20 13:42:23.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/7/check/DellEMC_PowerStore_iSCSI/9a81fb0/DellEMC_PowerStore_iSCSI/19 : SUCCESS in 1h 01m 56s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/7/check/DellEMC_PowerStore_FC/d9a4e8a/DellEMC_PowerStore_FC/20 : SUCCESS in 1h 02m 09s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"55f830ccef1319b99149011003a6ea3d098082a4","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-20 13:54:17.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"02f7de16dde447421bfa0a84c65e05e23f0e58d8","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-20 14:00:52.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/23f51e7dc3f8fb3ff6cc56baee953bf70ad35eac/index.html : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"960350b18662f743975014dcacb4cffd39af56ae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-20 14:02:19.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/6128121/6128121.html : FAILURE in 6m 19s","accounts_in_message":[],"_revision_number":8},{"id":"629f9481bbc897d08a4da9054b256c4f818dde81","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2020-07-20 14:02:27.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"0d5c167d80f811608d5a6890a190c652724de45f","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-20 14:11:50.000000000","message":"Patch Set 8:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":8},{"id":"3a3acef244abb6297722c5d974fd635b5e5bafaa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-20 14:30:56.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/8/check-not-vote/ext-cinder-zuul/c3df74f : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":8},{"id":"4e7ebe8bcabebee63004e73bffe50206429dd87a","author":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"date":"2020-07-20 14:37:09.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1711 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1919 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1745 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"37fde704e3dab39abd2c37cb9182d9e39aa3c326","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-20 14:45:26.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"d653031fcd8005338b792cb894d172cef982f0b7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-20 14:48:41.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/8/iscsi/ : FAILURE in 41m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/8/smb/ : FAILURE in 53m 46s","accounts_in_message":[],"_revision_number":8},{"id":"155b9c927ae822f73ff631b699474a92c1b27ea3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 14:50:28.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"d968bc29814ee589d4ddc7677db404d24090efaf","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 15:13:55.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"1a45b60d2d324bc1467a39a4ce99418d86578af9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-20 15:17:55.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/8/check/DellEMC_PowerStore_iSCSI/82b80de/DellEMC_PowerStore_iSCSI/20 : SUCCESS in 1h 05m 51s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/8/check/DellEMC_PowerStore_FC/0baddd4/DellEMC_PowerStore_FC/21 : SUCCESS in 1h 05m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"1431a34e53695b2d57c457864fcd287ee2a60c5f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-20 15:19:18.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f3a726c : FAILURE in 20m 57s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6974203 : FAILURE in 26m 29s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"39e8be999971b2145ffee26193bc939c8bb042e0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-20 15:47:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/8/check/dsvm-tempest-cisco-zonemanager-job/ee07ea5/ : FAILURE in 34m 59s","accounts_in_message":[],"_revision_number":8},{"id":"053c01de311750f31ea7225e18b25be656fb9583","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-20 16:01:39.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-8/ns5_nfs : SUCCESS in 2h 4m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-8/ns5_iscsi : SUCCESS in 2h 7m 4s","accounts_in_message":[],"_revision_number":8},{"id":"36adbea958594ccc9a02e8c93af10181d87e4cff","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 16:04:13.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"88bcd1ce4b6c56d0a2c28848ea92425464b46e0c","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-20 16:15:25.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/8/check/nimble-iscsi-driver-dsvm-volume/6741158 : FAILURE in 1h 06m 44s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/8/check/nimble-fc-driver-dsvm-volume/78e8377 : FAILURE in 1h 02m 24s","accounts_in_message":[],"_revision_number":8},{"id":"aa4b19cd73f848b0d4a9395d5d07b5595953d65f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:17:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e35daaa3b954ddbbb557fed45adbbfb : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d49b24846a541868be4c14a09dabffc : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01b21b64c5814a7880b06cbe2e83b73e : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef064dc19e7142719ccfb00f3b209d7e : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaeb0665b3bc46eba309572e90042ccd : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/37fbd761cb314eb98e340862ad792c19 : SUCCESS in 55m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06101dd098b44f51801ac6a12a4644c0 : SUCCESS in 59m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa68210fb2ab464c8d3453a1cd76e90f : SUCCESS in 7m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a6f51a8c3155452ea6e47c98a58f7658 : SUCCESS in 11m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c0c16d718cf44389c91c4608341c603 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b76befd28674c38b09b9940eb066856 : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a23b8e447c74dd19e80922a97dbead1 : SUCCESS in 7m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e47298c7e64544909d2c35e6e05498a4 : SUCCESS in 37m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca19cb2344f04cf3be77a39b48191ee4 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9fe2cd38246447ee998e6462de5d7fdd : SUCCESS in 1h 18m 46s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/80c1dd7ff6b143cbbfa5ce49a8d6e5c2 : FAILURE in 1h 44m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6478792b944e4687a27b10b0ecff6dab : SUCCESS in 1h 18m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8c44c1dedd454e6b866bbc6a42cbeede : SUCCESS in 1h 26m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0f3a302b4bfb40ea84b1a9bc8ac646a2 : SUCCESS in 1h 34m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/714d21975f3641aa8acfcc0086e41f80 : SUCCESS in 2h 15m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ead7a53ba00459bb4a69b7ef228ab86 : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":8},{"id":"a33969c253dc1649b6ce379fc464d2b86b688e45","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-20 17:51:42.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/8/check/DellEMC_XtremIO_FC/292ee04/DellEMC_XtremIO_FC/443 : SUCCESS in 1h 34m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/8/check/DellEMC_XtremIO_iSCSI/1f5db28/DellEMC_XtremIO_iSCSI/415 : SUCCESS in 1h 32m 45s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"680d357f3841914f423934547b6074d10353159e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 18:26:55.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"84f9efe96c9af94e6bdaec1fbcf31031624b2c19","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-20 18:48:54.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"cd642dbdd4715f73787aeba95c9700264f29e9bf","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-20 18:49:01.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c5c5388202fe35f7763777841a063e17770a2e05","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-20 19:40:04.000000000","message":"Patch Set 8:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"b464db9a26887db10663c4e6cfaa561f374f7d1a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 19:41:02.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4d20cea0f1ac7badf9cfb730ea503431424daaeb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-20 19:47:29.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/8/iscsi/ : FAILURE in 57m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/8/smb/ : FAILURE in 54m 17s","accounts_in_message":[],"_revision_number":8},{"id":"295e2f180de1df9f9881762fd9c9e9a4bdd77501","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-20 20:17:22.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/8/check/dsvm-tempest-cisco-zonemanager-job/49b923b/ : FAILURE in 32m 50s","accounts_in_message":[],"_revision_number":8},{"id":"b1d86a58393308e7905b524d427fddcc01c4b4bd","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-20 20:26:47.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4cdb60d : FAILURE in 20m 06s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/87899f5 : FAILURE in 20m 12s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"68c6dadecbc906050e7601b7a9c6fea940f8144d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-20 20:34:13.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"701f15d8af2178857dfb4d11e61124ffcaf6a7dd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-20 20:35:34.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/8/check-not-vote/ext-cinder-zuul/2e883db : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":8},{"id":"a30421fec64b963628100a2afa0a538c59b0a1e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 21:38:56.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebb45976ceed4570a6782cb6fd4c5b26 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41057aba21a4486f9c7a22361115d41d : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22333d4c10c142bc810dc7694be874f3 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/428c71b0c4134f79b50dd5b11f6635ec : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79d8204c6a74968b8704e5f2d689b04 : SUCCESS in 13m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/bd150c699b2348068bb373a405cfb153 : SUCCESS in 58m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ecd50769a5d84b95867eb6ba506c2b1f : SUCCESS in 1h 06m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53fa6a13d7fc4e35a63419d4376a4e89 : SUCCESS in 7m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1953a560e37e4a1e8e324717ff191a8f : SUCCESS in 12m 02s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc8bbc30a8de40e88d7a0e93e6408892 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef21340ea05c4fb591c76ba62b690d4e : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f8fc43c73cee49cf98f7605e68745619 : SUCCESS in 9m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e6038f2ae382410589422600174ff08e : SUCCESS in 37m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/394adeda0f794fb2a50d31b4b84f7efd : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/524bfe42c8294e728cc5396278376916 : SUCCESS in 1h 10m 06s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/838307ae218a485394a1ab63d930b1b2 : FAILURE in 1h 23m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/73f7eb4dff48409aba27930bf8740f0e : SUCCESS in 1h 03m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/331b540e4e5841d482729c541697fac8 : SUCCESS in 1h 27m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e42cdae0da704043a2e2fcde0039afbc : SUCCESS in 1h 25m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d6a1dc59e7a540fdb56858b49e9c528b : SUCCESS in 2h 43m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32f3eceb155c48b085bac553a6eb8dcc : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":8},{"id":"2811d26ded68dded1e9fc64ea46b1c203f447162","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-20 21:41:45.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"87145d0d8541539cd74996e289d754e7881d9491","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-21 06:34:46.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c0c76f217edf5b447aa4d67c9c6125f23d39dda4","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-21 09:26:39.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/741158/8/20200721151902/NEC-ISCSI : SUCCESS in 2h 30m 44s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/741158/8/20200721151902/NEC-FC : FAILURE in 3h 07m 35s","accounts_in_message":[],"_revision_number":8},{"id":"817c309a71b31c7ba43ebe32765b7372c2580fd6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-21 09:27:41.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"38803db152cd649d2c56942e4c8b177ab41f99ad","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-21 11:49:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"eb861a9ac6f7fef55d2501c2ced913ae47b3ec11","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-21 11:51:12.000000000","message":"Patch Set 9:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":9},{"id":"807a08e342cf94d06c6248cdb86996f0ca0012c6","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2020-07-21 11:57:50.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"88ffe1f2394888018b13a79ab5e2d71594745d49","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-21 12:55:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/430dd5c/430dd5c.html : POST_FAILURE in 7m 23s","accounts_in_message":[],"_revision_number":9},{"id":"fb942fb62d0e48d8948a63f1a3e31977471032c3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-21 13:03:54.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/9/check/DellEMC_PowerStore_iSCSI/12fd597/DellEMC_PowerStore_iSCSI/21 : SUCCESS in 1h 12m 26s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/9/check/DellEMC_PowerStore_FC/458663f/DellEMC_PowerStore_FC/22 : SUCCESS in 1h 06m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"4f63c7ef68e5a3c978dd284ebe4ca5c842b472f1","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-21 13:15:02.000000000","message":"Patch Set 9:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"e7b02283eca46d8624fa4fddc54ddfe9735a0be3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-21 13:16:11.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/9/iscsi/ : FAILURE in 1h 26m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/9/smb/ : FAILURE in 52m 28s","accounts_in_message":[],"_revision_number":9},{"id":"af73636a6afddfaf6276286c40fdcf74caebc3c5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-21 13:18:53.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"9e9476439936f2d12f1a7b995499ff12cd1941f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-21 13:59:39.000000000","message":"Patch Set 9:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-9/ns5_nfs : SUCCESS in 2h 9m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-9/ns5_iscsi : SUCCESS in 2h 8m 34s","accounts_in_message":[],"_revision_number":9},{"id":"07e584c8e76a926aa3edf02490becf9cadb18400","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-21 14:10:28.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/9/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/30d98e4 : FAILURE in 27m 16s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/9/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/7d2b750 : FAILURE in 21m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":9},{"id":"ff60a8270eb7c4c0b9b575620672e333a0f93a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 14:34:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3268f07c97e3408696e758e8b16118fc : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/668da2c8371f4156a2417adedce1ae98 : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb9979dc58814e899a55b75936e96710 : SUCCESS in 12m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7059b2100b842fd812f763aa2ea2301 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c4949a0f83f46d7bf6c4f05ec1d712b : SUCCESS in 14m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/c18c99f7560941c898d06508b58c0ba8 : SUCCESS in 1h 00m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e91d309f957461c9a0512fd9dea429a : SUCCESS in 1h 15m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4244283bf4ec4817ab2356b061303307 : SUCCESS in 8m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/61d0fcb544f54c139bdd55af284013c0 : SUCCESS in 12m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6a1750bf4a7443d88084e1205ebce41 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb2416b228c1428483b8bddbd3003f31 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/90a861e97ff34a8eb93e67b51780679e : SUCCESS in 9m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d32407b54162479294a4d68c438f0e4a : SUCCESS in 38m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9c6194044b14a5eaac2596a02b5b968 : SUCCESS in 7m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/321c5c3602f9491b8a24f6a699491deb : SUCCESS in 1h 11m 42s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/073cd7f55584450e8538f85c7e418923 : SUCCESS in 1h 38m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b6ec5d027e4d4642b7b593994f06a36f : SUCCESS in 1h 15m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a15346a23d640a398ccb64f059f3f21 : SUCCESS in 1h 45m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a5beb287ef2c48859b25d242a10122be : SUCCESS in 1h 25m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f3c3f31501274a07b871e4c7d317fd5c : SUCCESS in 2h 13m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a9a5078d86249b2ba72701b45e8ec4f : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":9},{"id":"b0b536c9a1682e9011bd97a62ca67c9d76514833","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-21 14:35:55.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/9/check/dsvm-tempest-cisco-zonemanager-job/e805171/ : FAILURE in 33m 01s","accounts_in_message":[],"_revision_number":9},{"id":"cb896d6cf051283a7cde9ede8fae4c25def91a9e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-21 15:00:38.000000000","message":"Patch Set 9:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-741158-9 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":9},{"id":"27007114c0296df58850b109982648deb8380cb1","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-21 15:28:18.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1720 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1927 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1753 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"5bee7c1d02e6ad2d7930b6a9828b1b3800b2fbd5","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-21 15:35:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"eb6afd935ea88799562ebf6d35c31ef0b40d5777","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-21 15:35:14.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/58/741158/9/check/synology-dsm-cinder-iscsi/075bfc9 : FAILURE in 28m 48s","accounts_in_message":[],"_revision_number":9},{"id":"6eaac6c1cdac447d7790f3795e355fe84dea791f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-21 15:36:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"19226c65b945aa8aaa3c2dff88ccbc3b73a5487d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-21 15:37:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fac1539ff28069ac825836cea924bb37f8f4efaf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-21 15:52:53.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)\n\nDocs included, third party CI is passing, and everything looks decent to me. A few very minor things noted for either follow up or if this needs to be updated for some other reason.","accounts_in_message":[],"_revision_number":9},{"id":"9ed1266f574c8590fb8282031204c6df899e41a2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-21 16:27:09.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"436f74620c6a76b470322ae241efcf7570910aaa","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-21 17:05:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/741158/9/upstream-solidfire-check/cinder-solidfire-iSCSI/ce44e12/ : SUCCESS in 2h 30m 25s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"17cc100d07fd908a406f7fe2800b8f5493e9cb7f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-21 17:16:22.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/9/check/DellEMC_XtremIO_FC/5178fb2/DellEMC_XtremIO_FC/453 : SUCCESS in 1h 31m 50s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/9/check/DellEMC_XtremIO_iSCSI/aa2f03a/DellEMC_XtremIO_iSCSI/425 : SUCCESS in 1h 34m 46s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"c0e207a0cbc3094a4068019de100f7baf971e00f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-21 18:02:38.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/58/741158/9/upstream-check/cinder-cDOT-iSCSI/0766a25/ : SUCCESS in 2h 52m 42s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/58/741158/9/upstream-check/cinder-cDOT-iSCSI-IPv6/b8fe53e/ : SUCCESS in 2h 25m 22s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/58/741158/9/upstream-check/cinder-cDOT-NFS/c852d55/ : FAILURE in 33m 29s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/58/741158/9/upstream-check/cinder-cDOT-NFS-IPv6/75bba5b/ : SUCCESS in 2h 57m 32s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/58/741158/9/upstream-check/cinder-cDOT-FCP/8442ecc/ : SUCCESS in 59m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"9d06494e7038179f2524f927960aa78d29bd29c6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 18:40:20.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/9/check-not-vote/ext-cinder-zuul/5cfe8d8 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":9},{"id":"53d3817983596bf19fd96ad590b9a50cbfb4479b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-07-21 19:35:39.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/58/741158/9/check/DellEMC_SC_FC/24b0141/DellEMC_SC_FC/202 : SUCCESS in 1h 03m 46s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/58/741158/9/check/DellEMC_SC_iSCSI/4ac6978/DellEMC_SC_iSCSI/175 : SUCCESS in 1h 34m 04s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"95acb373fd4d734c32c721f7271e18a8596c4f0c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-21 21:10:32.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/58/741158/9/check/DellEMC_PowerFlex/59b9c58/DellEMC_PowerFlex/282 : SUCCESS in 1h 56m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"fcac1d73345186812960c32bed3ea618e5c4f2af","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-21 21:19:33.000000000","message":"Patch Set 9:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/741158-9/HPMSA-iSCSI : FAILURE in 20:56.08\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/741158-9/HPMSA-FC : FAILURE in 15:45.41","accounts_in_message":[],"_revision_number":9},{"id":"6cb807d4b8c14fb4651e26f27f2cbaab86efa71a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-21 21:25:31.000000000","message":"Patch Set 9:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/741158-9/Seagate-iSCSI : SUCCESS in 47:55.28\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/741158-9/Seagate-FC : SUCCESS in 48:51.09","accounts_in_message":[],"_revision_number":9},{"id":"1c26698c15ecda6176d1f922129a45afc86ca9f8","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-21 22:08:12.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/9/check/nimble-iscsi-driver-dsvm-volume/c300d77 : FAILURE in 25m 38s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/9/check/nimble-fc-driver-dsvm-volume/3a7c5fa : FAILURE in 57m 14s","accounts_in_message":[],"_revision_number":9},{"id":"ba4193454bdf203dc2016d1fc29ca8823c0295c9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-21 22:36:10.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"305a5ee6e5173388121678cb7483a98f7e927b61","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-21 23:11:50.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/58/741158/9/check/dsvm-tempest-zadara-driver/ad07736 : SUCCESS in 1h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"43b455661dbc80a3f768778c7ff29c54f590daa8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-21 23:57:56.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/58/741158/9/check/DellEMC_VNX_iSCSI/ea13076/DellEMC_VNX_iSCSI/162 : SUCCESS in 2h 32m 04s\n- DellEMC_VNX_FC http://publiclogs.emc.com/58/741158/9/check/DellEMC_VNX_FC/0fbe7c9/DellEMC_VNX_FC/155 : SUCCESS in 1h 02m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"52cd9c87b22f92cdacd074180288d8acf123e3a5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 02:40:26.000000000","message":"Patch Set 9:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"fe909f2d346ff6d7ca475383be3047c3123ec096","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-22 03:53:33.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/0e9130846a3fbe2ddbf86b2b6ac8543e9fa1fb47/index.html : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"45334a74ce382c6f20b0395e4f307efeeea8b1b7","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-22 11:26:32.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"189a5f7cb5deca81fed120d06dc542db5f7c9ed8","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-22 11:27:50.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"31290b12cb844c0a637febe2c6eb6e5f6dd061b4","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-22 11:29:12.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"8bc719c0cc86d4a54a3b956e80151600e9811995","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-22 11:30:30.000000000","message":"Patch Set 11:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":11},{"id":"aa4b8e629e54c84fb4c8a08d5f3f9a1286c7b3a9","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-07-22 11:31:09.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/741158/9/20200722180004/NEC-FC : SUCCESS in 2h 15m 04s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/741158/9/20200722180004/NEC-ISCSI : SUCCESS in 2h 31m 02s","accounts_in_message":[],"_revision_number":9},{"id":"056eb0910356b0eb6bf21bd7802015cc58967803","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2020-07-22 11:35:28.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"682e6c0e35511136eb9a4c163ae23691db6e2774","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-22 11:36:31.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f54dafd/f54dafd.html : POST_FAILURE in 7m 32s","accounts_in_message":[],"_revision_number":11},{"id":"7add21314138dea53061501f7c4be61e7786d0ef","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2020-07-22 11:49:11.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"cedf5381c50bfe4257716eb1764577d50ed46420","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-22 12:25:29.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/11/iscsi/ : FAILURE in 56m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/11/smb/ : FAILURE in 50m 50s","accounts_in_message":[],"_revision_number":11},{"id":"320ec0e60c1744bc5908a111befc699ce11a8a8f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-22 12:34:54.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/11/check/DellEMC_PowerStore_iSCSI/f220676/DellEMC_PowerStore_iSCSI/22 : SUCCESS in 1h 04m 08s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/11/check/DellEMC_PowerStore_FC/81d3428/DellEMC_PowerStore_FC/23 : SUCCESS in 1h 02m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"2c4f21cd3fc748a84cc96966db92f8bfc051a4ac","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 12:52:07.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/11/check-not-vote/ext-cinder-zuul/eb84b13 : SUCCESS in 37m 18s","accounts_in_message":[],"_revision_number":11},{"id":"a9ef80f619b7b306e10191b3caac89a857248d9f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-22 12:54:44.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/11/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/6471c14 : FAILURE in 21m 24s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/11/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f371d5f : FAILURE in 20m 53s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":11},{"id":"1940e1db2a240f00402dc4ecfb2c099938c970d6","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-22 12:55:33.000000000","message":"Patch Set 11:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"e96a7c7a282eecff0f4920929f43cd92d9bae00d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-22 12:55:40.000000000","message":"Patch Set 10:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"75c753c06cc653fdd3cc4a32b21e153d761304b9","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-22 13:01:04.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/83cc2b4b08ea8304e609feee543fc59d7237605e/index.html : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"7089c78bb4bafbda849884ac34d265e70f22f7ec","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-22 13:37:55.000000000","message":"Patch Set 10:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-10/ns5_nfs : FAILURE in 2h 6m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-10/ns5_iscsi : SUCCESS in 2h 11m 2s","accounts_in_message":[],"_revision_number":10},{"id":"56db8f2a4f355b9ce9723afc65676fc0dcd49e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:42:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d655e489bdd5487881c61548cbffdb6c : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4161ece665f4b49819b379023ea1425 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f60d0998ba214239a1ffc2d8d12a64c0 : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b850114c074b60ba24d3719a5821dd : SUCCESS in 12m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af0be2898744e788f9b7512feb4e1ea : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/19e60608508a4406bc65fd58a8eff9f4 : SUCCESS in 42m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4b732b6c6f942fbae2655608f483f91 : SUCCESS in 1h 18m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86bfc083113f49baba80c2650e8d1ab6 : SUCCESS in 8m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d5502833c1ca4e21a7d7815bbcb660d9 : SUCCESS in 11m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/553ab25a1be2484baecadf231342ac27 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cca626c739b44352987c95d8012cc90c : SUCCESS in 6m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6333cca85c74fb2914b3b2f172de12c : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/61f4429914b5434c9124c46f22c4fb9d : SUCCESS in 44m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f543b360786e4b8c9d0a85b330434cab : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ea010643bee415f9dc0a560bc498baa : SUCCESS in 1h 04m 50s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8143ca20eb044277ab2022faeceea52d : SUCCESS in 1h 50m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/22119f27c5164acb8d314b2dc61b8dce : SUCCESS in 1h 13m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9ac8d82fa3d04fae9e381abddb8d37cc : SUCCESS in 1h 07m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5c2cd827ff94408faa1541c0f6f4da98 : SUCCESS in 1h 15m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b742322d81044788eb2457ea9b2be19 : SUCCESS in 1h 56m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f0b7ed2c77b4f8c9c425050eb9f4583 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":11},{"id":"c4d2310a79e4c2019e5b75f28094072fadcfc935","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 13:46:10.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1\n\nJust minor updates, and Zuul and third party CI were happy with this prior. Going to assume everything passes again.","accounts_in_message":[],"_revision_number":11},{"id":"cd7277c70350b898df4f65444d86aeb153bc4b4e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-22 13:46:25.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/11/check/dsvm-tempest-cisco-zonemanager-job/0741c8f/ : FAILURE in 33m 16s","accounts_in_message":[],"_revision_number":11},{"id":"faf935d5982d4a52dd9cffac7b2e86fa10de5d8c","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-07-22 13:49:47.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"f179a85e8bb970947ab1c4f1fb2421cebd368334","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-22 13:52:51.000000000","message":"Patch Set 11:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-11/ns5_nfs : SUCCESS in 2h 24m 44s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-11/ns5_iscsi : SUCCESS in 1h 56m 6s","accounts_in_message":[],"_revision_number":11},{"id":"5103f031e839c1016c30b47a61518a745d8d456b","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-22 14:16:53.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1739 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1941 : SUCCESS\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1767 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"e08017a27db2c7030eebd06f41f7e0592b890ce8","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-07-22 14:24:48.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)\n\nHi Ivan,\n\nThe support matrix is telling that this driver supports the online extend feature, but the CI is not testing it.","accounts_in_message":[],"_revision_number":11},{"id":"c1735da80d88a36bacbf5a54d378bea1ab4803ca","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-22 14:28:52.000000000","message":"Patch Set 11:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-741158-11 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":11},{"id":"f6618d618d013c90ae80973fdb92ec3bbcc0101c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-22 14:38:53.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/741158/11/upstream-solidfire-check/cinder-solidfire-iSCSI/0f5671a/ : SUCCESS in 54m 44s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":11},{"id":"bf0d99485667b82c2f0fe946290e283f12c9679e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 14:46:27.000000000","message":"Patch Set 10:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"c15c215e128e4f161a01fcc41cee223f24ca7d74","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 14:47:46.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"b3a72d218c6d85a89f3bace640d4ab11c7cff307","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-22 14:59:13.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/58/741158/11/check/synology-dsm-cinder-iscsi/9063775 : FAILURE in 28m 24s","accounts_in_message":[],"_revision_number":11},{"id":"d34933f96621eaebd12a3f51e2556ad2a802109d","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-22 15:14:06.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1740 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"767be932e682feccb82cba8b0f04d45342d5554e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 15:28:09.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"102e8d762028857a1f950d2b656e7fef9947bb05","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-22 16:32:08.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"90c874e90d1e749c855dc5a7b69174d6897fce73","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-22 16:45:03.000000000","message":"Patch Set 11:\n\n\u003e (1 comment)\n \u003e \n \u003e Hi Ivan,\n \u003e \n \u003e The support matrix is telling that this driver supports the online\n \u003e extend feature, but the CI is not testing it.\n\nextend_attached_volume test is failing for both iSCSI and FC because of this bug https://bugs.launchpad.net/os-brick/+bug/1793259.\nPowerStore allows to use luns greater than 255 but sg_scan can\u0027t return the right lun.\nWe can limit attached luns to 255 but I think this way of resolving issue is wrong.","accounts_in_message":[],"_revision_number":11},{"id":"a0b79c0281a3f8b894350065968c1889a25a7cc6","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-07-22 17:00:46.000000000","message":"Patch Set 11:\n\nIn this case, you may report that the feature is missing for now, and update the support matrix once the bug is fixed.\n\nOtherwise your customers might get errors trying to extend an attached volume, or even worse, have a wrong volume being extended.","accounts_in_message":[],"_revision_number":11},{"id":"229ab420e7d928b452d529682dde8e74f3712ef5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-22 17:10:27.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/58/741158/11/upstream-check/cinder-cDOT-iSCSI/b12cb39/ : SUCCESS in 2h 11m 31s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/58/741158/11/upstream-check/cinder-cDOT-iSCSI-IPv6/f7f1cad/ : SUCCESS in 2h 27m 20s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/58/741158/11/upstream-check/cinder-cDOT-NFS/067e04a/ : SUCCESS in 2h 03m 57s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/58/741158/11/upstream-check/cinder-cDOT-NFS-IPv6/1922279/ : SUCCESS in 2h 36m 22s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/58/741158/11/upstream-check/cinder-cDOT-FCP/75ca050/ : FAILURE in 2h 28m 57s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"95be168e2c45d02ef9959e6e5fd797630df00c89","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-07-22 18:14:26.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/58/741158/11/check/DellEMC_SC_FC/b285450/DellEMC_SC_FC/212 : FAILURE in 1h 06m 26s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/58/741158/11/check/DellEMC_SC_iSCSI/b07ffea/DellEMC_SC_iSCSI/185 : FAILURE in 1h 30m 15s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":11},{"id":"13176de0cecb071bb68db28f05fcf294fc8aec0a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-22 19:27:08.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/58/741158/11/check/DellEMC_PowerFlex/1fe0041/DellEMC_PowerFlex/292 : SUCCESS in 1h 55m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"663baa58cd9498deeecc0d702e96a07bd6a598cd","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-22 19:40:57.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"75437b0501089f5552ba2fff674eacaedf1c9c71","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-07-22 19:42:08.000000000","message":"Patch Set 12:\n\n\u003e In this case, you may report that the feature is missing for now,\n \u003e and update the support matrix once the bug is fixed.\n \u003e \n \u003e Otherwise your customers might get errors trying to extend an\n \u003e attached volume, or even worse, have a wrong volume being extended.\n\nDone","accounts_in_message":[],"_revision_number":12},{"id":"e20788af188213d020690a56406852c439f0db3d","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-07-22 19:52:28.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"97ee4f5e482bdbd204efd3de9b9abfedd10098bc","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-22 20:11:05.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/12/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c7eba3e : FAILURE in 20m 11s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/12/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/2fa961a : FAILURE in 19m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":12},{"id":"a8bd49bcbaeb8c1d739958e4a19739823dd8cde7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 20:17:47.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/12/check-not-vote/ext-cinder-zuul/9f1afb6 : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":12},{"id":"ea8108b829c5284516fc03f1a40232d1daa41a4b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-22 20:18:52.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/12/check/dsvm-tempest-cisco-zonemanager-job/d32793e/ : FAILURE in 32m 50s","accounts_in_message":[],"_revision_number":12},{"id":"ee0ef3415806f10f4e691e79ab92088e723444cb","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-22 20:31:52.000000000","message":"Patch Set 12:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"96e67ac2a651f72fda7bc2fe62efd234ab4898df","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-07-22 20:34:51.000000000","message":"Patch Set 12: Code-Review+1\n\nLGTM, thank you Ivan!","accounts_in_message":[],"_revision_number":12},{"id":"33b268a7f4e1c90a9ab8c1c07bdec24b9541bc85","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-22 20:40:23.000000000","message":"Patch Set 11:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/741158-11/Seagate-iSCSI : SUCCESS in 54:46.73\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/741158-11/Seagate-FC : SUCCESS in 55:57.05","accounts_in_message":[],"_revision_number":11},{"id":"052dca5acbc4e5efcd310f03aeff652cac0cf6c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 20:42:32.000000000","message":"Patch Set 12: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"9d9da336e751c1a183388cc2e376e7f87d5f812d","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2020-07-22 20:42:42.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"1f0eb90304539ec84000fe0d23dc2081827306ea","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-22 20:45:43.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/12/iscsi/ : FAILURE in 1h 03m 57s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/12/smb/ : FAILURE in 54m 45s","accounts_in_message":[],"_revision_number":12},{"id":"4688161d15bb54f99c41fd112b3b3cd3f6fd543c","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-22 20:47:46.000000000","message":"Patch Set 11:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/741158-11/HPMSA-iSCSI : SUCCESS in 45:36.72\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/741158-11/HPMSA-FC : SUCCESS in 44:13.88","accounts_in_message":[],"_revision_number":11},{"id":"15aaf8d947163aee10161efcf5d7e2013c517aec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-07-22 20:50:53.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/741158/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a76ce47/a76ce47.html : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":12},{"id":"043af2a421e28e92ab451fb923688fdb66acf8eb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 21:02:34.000000000","message":"Patch Set 11:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"6db5878d39e3284d9decce6679992777c91be92c","author":{"_account_id":22126,"name":"Huawei FusionStorage CI","email":"294958976@qq.com","username":"hustnw","tags":["SERVICE_USER"]},"date":"2020-07-22 21:12:53.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* fusionstorage-dsvm-tempest http://116.66.187.79/fusionstorage-dsvm-tempest/1745 : FAILURE\n\n* huawei-iscsi-dsvm-tempest http://116.66.187.79/huawei-iscsi-dsvm-tempest/1946 : FAILURE\n\n* huawei-fc-dsvm-tempest http://116.66.187.79/huawei-fc-dsvm-tempest/1772 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"5a34c76308abc9164757c46e4f5c2e9cdb8a01fd","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-22 21:15:46.000000000","message":"Patch Set 11:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/741158-11/Lenovo-iSCSI : SUCCESS in 51:05.14\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/741158-11/Lenovo-FC : SUCCESS in 46:00.66","accounts_in_message":[],"_revision_number":11},{"id":"dc93d668385a3271081bd6f8dc036855c25fb9da","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-07-22 21:47:38.000000000","message":"Patch Set 12:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-741158-12/ns5_nfs : SUCCESS in 2h 4m 35s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-741158-12/ns5_iscsi : SUCCESS in 2h 6m 17s","accounts_in_message":[],"_revision_number":12},{"id":"066202054ba10a535ca2982a8dd4bc1815ae7902","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-22 22:26:36.000000000","message":"Patch Set 12:\n\nWaiting for Zuul and the third-party CI.  Thanks to Sean and Lucio for all the reviews on this new driver.","accounts_in_message":[],"_revision_number":12},{"id":"953e007e0938107329920f84d93b26226651b563","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 22:41:29.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"7efb85cecfe464f4b0ae4b86acd4d1a1506c1f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 22:58:07.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bf861ae27434cb682465a423c6b8bbe : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7412c81ca14335855b8c2cfba01280 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/222b4072caac4a29908b9f1555783a82 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b78246d8cb9541089fee405b22f3bcc1 : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb0a7a791324e509adce3162b0e6a7f : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/be40853283a44c3787740bda0a39c423 : SUCCESS in 58m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/15525ac1e0d74b8bb47356a6f4651002 : SUCCESS in 1h 21m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bf6fb31549a4141a1b6074c1d4a65cc : SUCCESS in 9m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/480603c0f6344579aea54b70c4925d6d : SUCCESS in 12m 05s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/feea407a66434328a0cae84107bbb3c2 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/135e520bf0244786bd41eccda23c88fd : SUCCESS in 5m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/edf72fe53ca344b1a1f14d5d7034b7c9 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8fd46442a91540dc8f5517d7bdc775f4 : SUCCESS in 44m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c7502d968684a72b0e656210eaa4364 : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/69f7d5be444e41468524a7d054738a9a : SUCCESS in 1h 46m 06s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89adeab58e594264bb7e69e426d196e6 : SUCCESS in 1h 11m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4a2d96d8d9f64b079f00bc59e2a77564 : SUCCESS in 1h 09m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/81ffcf411e0147e0b849bbaf86109017 : SUCCESS in 1h 18m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/accb1a32ec514913b61636f5d94a77d2 : SUCCESS in 1h 26m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c811235b8594d7e9dd8990bc2bb05bf : SUCCESS in 2h 05m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cda997926b24df2add0cb7afedb5dc5 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":12},{"id":"9236e49a9d570a5c0f0f6f17f9f9ecdd00f577c1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-22 23:13:57.000000000","message":"Patch Set 12: Code-Review+1\n\nOK, now just waiting for PowerStore CI.","accounts_in_message":[],"_revision_number":12},{"id":"d59c54323f11a5c48885bc80a8b7796aeb19c41f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-22 23:24:02.000000000","message":"Patch Set 12:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-741158-12 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":12},{"id":"f6c02060cfd535ed6933b498aaaf0326a3f84415","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-22 23:26:30.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/58/741158/12/check/synology-dsm-cinder-iscsi/9e76b23 : FAILURE in 27m 49s","accounts_in_message":[],"_revision_number":12},{"id":"ffaa9e0a1d65b43bb11a20e5f54f02740569f95e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-22 23:43:23.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"f51b8dd20ad1dc30cc2f0c4d1866c0af87a64914","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-23 00:51:45.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/741158/12/upstream-solidfire-check/cinder-solidfire-iSCSI/8017fbf/ : SUCCESS in 1h 52m 26s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"772bf25bcc8a6c7543b3d3958bfad9b400427d45","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-23 01:44:30.000000000","message":"Patch Set 12:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":12},{"id":"43ed65be46a96d0ae1173e1609209704fb392fe0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-23 02:51:32.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/741158/12/check/DellEMC_PowerStore_iSCSI/0ddc12b/DellEMC_PowerStore_iSCSI/23 : SUCCESS in 1h 06m 44s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/741158/12/check/DellEMC_PowerStore_FC/a1ed1ff/DellEMC_PowerStore_FC/24 : SUCCESS in 1h 02m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"7fedb610fc516f5c5ef95224b392b91037a5cec9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-23 02:53:07.000000000","message":"Patch Set 12: Code-Review+1\n\nCI is happy now.","accounts_in_message":[],"_revision_number":12},{"id":"76701677d47b80f88a179853925b7e5d44131c18","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-23 02:53:21.000000000","message":"Removed Code-Review+1 by DellEMC PowerStore CI \u003cdellemc.powerstore.ci@dell.com\u003e\n","accounts_in_message":[],"_revision_number":12},{"id":"e33c4f564f3ed46d4c36db398148516d6f64ab03","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-07-23 02:55:40.000000000","message":"Patch Set 12:\n\nCI is happy now","accounts_in_message":[],"_revision_number":12},{"id":"d5671b1fd936c179bc064f41583267bca700f686","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-07-23 03:38:37.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/517cb6448b613b888476f5da75ae4ad28f60f744/index.html : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"673159d9601d32caab924ef93873aaaac186d012","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 03:46:29.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"7f7d87adcca2bfbbd919ba12e8e11024de203415","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-23 06:45:56.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/58/741158/12/check/dsvm-tempest-zadara-driver/07d66d3 : FAILURE in 1h 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"220a178e82d34834fcba2b72530e24dbc41066da","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-23 07:06:21.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/58/741158/12/check/DellEMC_PowerFlex/3fdb154/DellEMC_PowerFlex/298 : SUCCESS in 1h 57m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"54efc52525f20a4b476320b46787411c5daceac2","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 07:42:30.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"eb76e632915f01a49b72c97ecfa9cae09be4f655","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-23 09:06:38.000000000","message":"Patch Set 12:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/741158-12/HPMSA-iSCSI : SUCCESS in 46:57.00\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/741158-12/HPMSA-FC : SUCCESS in 46:12.75","accounts_in_message":[],"_revision_number":12},{"id":"cbdb32519f6505230b1e1727e7830025575464ec","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-23 09:20:27.000000000","message":"Patch Set 12:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/741158-12/Seagate-iSCSI : SUCCESS in 44:17.60\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/741158-12/Seagate-FC : SUCCESS in 46:46.73","accounts_in_message":[],"_revision_number":12},{"id":"a145cbaadc42affd77c71319d32318c37517e8b3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-23 09:56:43.000000000","message":"Patch Set 12:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/741158-12/Lenovo-iSCSI : SUCCESS in 46:33.64\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/741158-12/Lenovo-FC : SUCCESS in 42:03.82","accounts_in_message":[],"_revision_number":12},{"id":"cb55909ba5472e2868b224ce78135f40ffb10156","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-23 10:44:12.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/741158/12/check/tempest-dsvm-full-bionic/2736dde/ : SUCCESS in 1h 34m 47s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/741158/12/check/devstack-plugin-ceph-dsvm-bionic/d2bd1f5/ : SUCCESS in 1h 31m 21s","accounts_in_message":[],"_revision_number":12},{"id":"4c59c44e70460ba64d47725403a91612e440221f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 12:00:16.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"06c595d099a113859ce3f88ddd3242c1e39891f5","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-07-23 12:01:45.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/58/741158/12/check/DellEMC_SC_FC/41151d6/DellEMC_SC_FC/218 : SUCCESS in 1h 08m 09s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/58/741158/12/check/DellEMC_SC_iSCSI/dd980c3/DellEMC_SC_iSCSI/191 : FAILURE in 1h 30m 10s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":12},{"id":"bdcf68d624ab36528a864b076dabd09d7ebcee04","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-23 12:35:49.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nEverything looks in order here.  Thanks for contributing a new volume driver to Cinder!","accounts_in_message":[],"_revision_number":12},{"id":"d2555fe5554d6acfb8fd49826e76b90f92e573ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 12:36:21.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"95689f547d51d32a8bba9bc6cbc087a3a598fc9c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 13:14:12.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"605b1015788c9bba40ab2bf61ed3d42ecb5508a2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-23 15:29:00.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/12/check/DellEMC_XtremIO_FC/3a058eb/DellEMC_XtremIO_FC/479 : SUCCESS in 1h 33m 58s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/12/check/DellEMC_XtremIO_iSCSI/714fe8f/DellEMC_XtremIO_iSCSI/451 : SUCCESS in 1h 34m 00s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"9344af2a38204847c5a2e67ea76d625b35d73abb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 16:05:26.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"0d8bc93ce15d421e06be70374565062848dba222","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 16:41:29.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86219b5a150e4ea99906660034569f7d : FAILURE in 40m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60d7d3515ed44428f7b5bf01085b303 : SUCCESS in 10m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ff7f41d948c4d6883e7f0d86c0fb4d9 : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04335f5eda094d268ea45b36dcb96826 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d853ffe119934a12b86e136983e6da60 : SUCCESS in 18m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2376c3076b1d46b9b20eccb3a08130ce : SUCCESS in 1h 04m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76cf83b498bf4408aeb0e4b70c46c020 : SUCCESS in 1h 14m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef63f631a6ea4ab0a719acbead6c371b : SUCCESS in 9m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2cb8791d59ab4050b183fbec08b24b13 : SUCCESS in 1h 15m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8c3a8a5c6ec2425daa12ded1d473f070 : SUCCESS in 1h 21m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a647bc365c94b2ea646238a70d9124b : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":12},{"id":"d0b1b6340c4ccb7b82f35290a2538d5bdf4e5dfc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-23 16:46:35.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"e6af4a7fea3e3886869982c00a8105fdbd43acb5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-23 17:07:22.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/12/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/5d70a33 : FAILURE in 20m 29s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/12/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/b6b7f93 : FAILURE in 19m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":12},{"id":"037befafd98bd9befd86dbc53f1a6e8a9584817b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 17:22:39.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"e97a986f61b644132951140dd8029ee4af7adc23","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-23 17:24:28.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/12/check-not-vote/ext-cinder-zuul/d07c63b : SUCCESS in 37m 19s","accounts_in_message":[],"_revision_number":12},{"id":"17d062e73397fa537fcb73ca38d8cd668201fde7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-23 17:24:30.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/12/check/dsvm-tempest-cisco-zonemanager-job/343ea9f/ : FAILURE in 32m 45s","accounts_in_message":[],"_revision_number":12},{"id":"4d369f88156ee801edd99895d997536e0afa3b20","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-23 17:37:46.000000000","message":"Patch Set 12:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"98763cd4e7d81bf2a4d45de9f87fb5c462c7c6e3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-23 18:01:32.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/58/741158/12/check/dsvm-tempest-zadara-driver/b80d3bf : SUCCESS in 1h 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"13b96e43a23ba3459583860245be1d23cb01433b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-23 18:14:35.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/12/iscsi/ : FAILURE in 1h 27m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/12/smb/ : FAILURE in 51m 20s","accounts_in_message":[],"_revision_number":12},{"id":"a243f9551a61003b4c0d73c1e9b33826da5bbc6b","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-07-23 18:16:24.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1358/index.html : SUCCESS in 1h 19m 53s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1506/index.html : FAILURE in 1h 29m 32s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1529/index.html : FAILURE in 54m 06s","accounts_in_message":[],"_revision_number":12},{"id":"a17ce86b6173e6d1a2cab5a27f2069e073440875","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-07-23 18:27:43.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/12/check/nimble-iscsi-driver-dsvm-volume/2058fe9 : SUCCESS in 2h 29m 21s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/58/741158/12/check/nimble-fc-driver-dsvm-volume/0a3c2ef : SUCCESS in 1h 55m 59s","accounts_in_message":[],"_revision_number":12},{"id":"8b82b32325c304fdcca0fb9612c46932a9eb1ffc","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 18:47:38.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"313d06866bb337f2b048bb44550f9f2798623f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 19:10:28.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18c072427e8343048d443445ba473ecc : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af34699a1924886ae9e6432337464b3 : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be7c03d1ff6646b1b460e9153fba804a : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dd2e8cd6bc8457b891b8de314de2922 : SUCCESS in 12m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8748d2da2c4c12a70f38c47603e8a3 : POST_FAILURE in 17m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/30b28a8bfba94a138c7fbfc293be614d : SUCCESS in 1h 18m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d4f1f1f51ea4c0cb6f470cb141b1756 : SUCCESS in 50m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c6a8bf5c5f74fb393fdb499478e2776 : SUCCESS in 9m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7504a76cfac44a14b5eef2fd04077e4d : SUCCESS in 12m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/47263508e49d4fe08824489ca49305a6 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb863aa9be8d4498b245d4103f88dc5a : SUCCESS in 10m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5f6ddd9c3258468f88daca1d8511f224 : SUCCESS in 9m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c9520418384342188c432f02229e407e : SUCCESS in 37m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e77cb1d0592c4b7bb4027535b961048c : SUCCESS in 10m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7fe5ae9a9439487a827d86df1d883ffb : SUCCESS in 50m 40s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b2778d5e795047f08ef9a4d974c5f489 : SUCCESS in 1h 32m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/78fbf25165154b4596d20112c9b33c5a : SUCCESS in 1h 13m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/77cef020230e45c0b371201b7d6a47a4 : SUCCESS in 1h 22m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/488b4ed6f2764ee1a64b188216d09946 : SUCCESS in 1h 32m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7ed6e1e7bab4bb182b77b97bc80a636 : SUCCESS in 2h 05m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6021845ed7864a49915dba40267d7cc1 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":12},{"id":"b9e545a2177c126f9962c77a602127576e6c5e53","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-23 19:11:21.000000000","message":"Patch Set 12:\n\nrecheck\n\nNetwork glitch after docs job completed caused upload failure.","accounts_in_message":[],"_revision_number":12},{"id":"2224d58d31ba411f2ff2b4194fcce49d315ca6c0","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-07-23 19:41:57.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"95c9e99186748438a301dc0633529b0efe783347","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-23 19:47:39.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/12/check-not-vote/ext-cinder-zuul/7106410 : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":12},{"id":"d494e22ca256120601bbd5203a81f88df8332912","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-23 19:48:35.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/12/check/dsvm-tempest-cisco-zonemanager-job/e5b7c69/ : FAILURE in 32m 05s","accounts_in_message":[],"_revision_number":12},{"id":"f96cbdbd27e0eb53c00c6177bdcc4f33f0efd8fe","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-23 20:01:35.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/58/741158/12/check/DellEMC_VNX_iSCSI/4b4e298/DellEMC_VNX_iSCSI/177 : SUCCESS in 2h 36m 41s\n- DellEMC_VNX_FC http://publiclogs.emc.com/58/741158/12/check/DellEMC_VNX_FC/5ba82fc/DellEMC_VNX_FC/171 : SUCCESS in 1h 05m 26s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":12},{"id":"204dde4d8740628633d9ae38a684befc49039e1b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-23 20:02:17.000000000","message":"Patch Set 12:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"be8073d530f5a228b6c680c27319bc1fdd65a3db","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-23 20:14:35.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/12/iscsi/ : FAILURE in 1h 02m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/12/smb/ : FAILURE in 58m 24s","accounts_in_message":[],"_revision_number":12},{"id":"818c5ccdceaae87558fb05fa5bd021f350ac1a4c","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-07-23 20:29:17.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1360/index.html : SUCCESS in 1h 17m 37s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1508/index.html : FAILURE in 1h 04m 32s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1531/index.html : FAILURE in 51m 19s","accounts_in_message":[],"_revision_number":12},{"id":"0d88bfe3f6d98879ea5e038f35def612e6007460","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-23 20:52:38.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/58/741158/12/check/dsvm-tempest-zadara-driver/0cc66f9 : FAILURE in 1h 03m 45s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6013570894c159c56cc64dfaffc3ddd99c3e6db7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 21:21:41.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/148e8790791b421fa319cdaf944d5e46 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c7beb8b3ed412eaed45ae09a805d9b : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4d9660b3fc64efdbd9cc711d9158a74 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7fc29076ec8445894b8d34da4f49834 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631a343e3acb4fdd9d80fda43ed253d2 : SUCCESS in 14m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/e9788dfa87874025bf4eafd3b4ad8343 : SUCCESS in 1h 10m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69fad66e22504d5d82ba5afe9b0018e3 : SUCCESS in 1h 19m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89b8b6ddd758429ab80f2f4e17331899 : SUCCESS in 8m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a879c3195f1e4005a4eaefedc3953f25 : SUCCESS in 14m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ea3506740e545178408be082d894db6 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa1f5155dad640fd95899e7a96697246 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b3b4bccc7c34ab0a3f6189f98499507 : SUCCESS in 8m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e426a8d179c4d1da69093bc103ce581 : SUCCESS in 1h 01m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b2a12470c9a54b6faef13684dde192ab : SUCCESS in 8m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ad9a847d55849d7a3233f78b6f96842 : SUCCESS in 59m 23s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/491f81ab82fa46b3832c29b5cb409782 : FAILURE in 1h 40m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/90315f7421f84e399040214b05f8c35b : SUCCESS in 1h 23m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/18656239982348ce8345c60997100763 : SUCCESS in 1h 34m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/92887fb3cb3f41fc8d53d93cd2a0ba47 : SUCCESS in 1h 41m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/efa9c1e595954f568db47e78bc9b786c : FAILURE in 2h 05m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4535ff0f0ff04db9bf39d017d788786d : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":12},{"id":"93fb3351aeeb21c27d8f05cd6eecef94bc045278","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 21:25:03.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"165245e412dca34439831bf36ffee9d9253e2f09","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-23 21:27:48.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"d19b804118114f8c69dcdd6fd5d2f68eec941068","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-23 21:55:19.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/58/741158/12/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/da22122 : FAILURE in 19m 53s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/58/741158/12/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/1ca44c0 : FAILURE in 27m 13s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":12},{"id":"17d5306a5a45adb560f9e8e43849f1da28ebdefb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-23 22:02:20.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/58/741158/12/check-not-vote/ext-cinder-zuul/5085b56 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":12},{"id":"2a3f92b7a6b64b3e751f628d515591d93ef872ce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-07-23 22:05:08.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/741158/12/check/dsvm-tempest-cisco-zonemanager-job/39de227/ : FAILURE in 32m 14s","accounts_in_message":[],"_revision_number":12},{"id":"025448ace88576b1612a15a09ff2d54f054ef18e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 22:10:37.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"01e941b33e0833acc5f667a0658c76129132d59b","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-07-23 22:18:46.000000000","message":"Patch Set 12:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"2e9243f07fb5223097de15f040c39d22c8bc590f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-07-23 22:28:21.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/741158/12/iscsi/ : FAILURE in 1h 00m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/741158/12/smb/ : FAILURE in 26m 31s","accounts_in_message":[],"_revision_number":12},{"id":"358a71c12c5f2d92521abb519f7f0fd4ee2e4cc2","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-07-23 22:42:39.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1361/index.html : SUCCESS in 1h 14m 35s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1509/index.html : FAILURE in 56m 31s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1532/index.html : FAILURE in 49m 49s","accounts_in_message":[],"_revision_number":12},{"id":"714cbe4f246db7ca853f44bc775a2d9495b4d480","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-23 22:45:00.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/741158/12/check/tempest-dsvm-full-bionic/aa2206f/ : SUCCESS in 1h 44m 15s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/741158/12/check/devstack-plugin-ceph-dsvm-bionic/84fd279/ : SUCCESS in 1h 29m 24s","accounts_in_message":[],"_revision_number":12},{"id":"093c701fc48f1b0ce1e70ff463adbed0a3946819","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-07-23 22:46:15.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/58/741158/12/check/dsvm-tempest-zadara-driver/b7fc664 : FAILURE in 1h 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e5627b1804c6ae2fe95340e004ac71331d4760d0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-23 23:51:24.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"de08bb9fa4750da21342eb5a440640d914271d85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 23:53:45.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a5b21efc3ff45938e8aea039e3ad328 : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5502ec1ef85467b81d967371fb74096 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8abb379b27b24e61851199fbe19f3e40 : SUCCESS in 12m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30350238bbc34430ab6a5228537cd14e : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84dcc8071c24713a9228eb5d46adbe7 : SUCCESS in 14m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/b109e45d22a14e9aa27dbf0f58cc76fb : SUCCESS in 55m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c847120da0e244caac6146bed131b9e8 : SUCCESS in 1h 00m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9107f8aebe9541e1a7d5e105573f2775 : SUCCESS in 7m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/93d7682aeea74a74bb086574def81c7a : SUCCESS in 12m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/248d610d31644bd2988c4dc2623bbb9c : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ae115527983442e9769f0f77081c7f4 : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d4c15b011744eb1ba623c0ca73a2bfb : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/270f210ab9df4d2c80311fc403048637 : SUCCESS in 40m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c781ee1b28747b9a8f2da39f4be185a : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6f27dfae6800453ea0729adb5941ce65 : SUCCESS in 1h 02m 36s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/064df843f71b4c67b62e713859f011b2 : SUCCESS in 1h 27m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/78e5979f2fdc4246a3ef97a09efbe69c : SUCCESS in 1h 12m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bdda62bdfb0f4cdf94234891f70385b5 : SUCCESS in 1h 26m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ed14f5d886a04bf3ac57c94d136831d9 : SUCCESS in 1h 20m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf6a436a669743768e18db23febdca7f : SUCCESS in 2h 23m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41e727bb4d274ab8942b1df3775cf7df : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":12},{"id":"a0545aa53657766989740c5f2ee15d7ecdd5775f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 23:54:16.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"956c54c7551192b90383225829c10303de7df315","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-24 00:20:27.000000000","message":"Patch Set 12:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-741158-12 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":12},{"id":"5926b12d29179a506e0365c3d01ef6e3e04c6d3b","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-07-24 00:21:03.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/58/741158/12/check/synology-dsm-cinder-iscsi/0752fd0 : FAILURE in 26m 47s","accounts_in_message":[],"_revision_number":12},{"id":"2cd3f5b928cf6cce7991fbc7669ab58a9b0ec824","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-24 00:40:15.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/58/741158/12/check/DellEMC_PowerFlex/a830fe6/DellEMC_PowerFlex/301 : FAILURE in 46m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"5489263f2bab73864d6540819139258186c8735b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-24 00:54:09.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"b38f644865e31d40420ca40062ddb7cd2fce71a0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-24 01:28:08.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/741158/12/check/DellEMC_XtremIO_FC/b91d120/DellEMC_XtremIO_FC/483 : SUCCESS in 1h 34m 03s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/741158/12/check/DellEMC_XtremIO_iSCSI/0c24021/DellEMC_XtremIO_iSCSI/456 : SUCCESS in 1h 33m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"73ea07e06fc3c35d8d3c1043ee6d59f1a4f89a20","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-07-24 01:29:25.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/58/741158/12/check/DellEMC_SC_FC/f214813/DellEMC_SC_FC/221 : FAILURE in 1h 16m 37s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/58/741158/12/check/DellEMC_SC_iSCSI/f658128/DellEMC_SC_iSCSI/195 : SUCCESS in 1h 35m 21s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":12},{"id":"181455ea76dff60d4fb2c85a071884527b97533d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-24 01:29:49.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/741158/12/upstream-solidfire-check/cinder-solidfire-iSCSI/d74027e/ : FAILURE in 1h 35m 15s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"f0da3cc6b25bfc55a0f033c073303b55f845e177","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-24 01:33:56.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/58/741158/12/upstream-check/cinder-cDOT-iSCSI/0bed5b2/ : SUCCESS in 37m 19s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/58/741158/12/upstream-check/cinder-cDOT-iSCSI-IPv6/51fa31f/ : SUCCESS in 1h 32m 32s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/58/741158/12/upstream-check/cinder-cDOT-NFS/cd0982e/ : SUCCESS in 39m 27s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/58/741158/12/upstream-check/cinder-cDOT-NFS-IPv6/21be3b7/ : SUCCESS in 1h 39m 39s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/58/741158/12/upstream-check/cinder-cDOT-FCP/4f8347f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"3ad409a31bad9527bbfc4d4fb7cf56b99e8a9eb0","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-07-24 01:47:11.000000000","message":"Patch Set 12:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/741158-12/Seagate-iSCSI : SUCCESS in 41:40.56\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/741158-12/Seagate-FC : SUCCESS in 39:47.01","accounts_in_message":[],"_revision_number":12},{"id":"746a92a9d46eb2fb1313f15c5ae89a1a8376183f","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-07-24 02:03:20.000000000","message":"Patch Set 12:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/741158-12/HPMSA-iSCSI : SUCCESS in 48:33.95\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/741158-12/HPMSA-FC : SUCCESS in 46:06.55","accounts_in_message":[],"_revision_number":12},{"id":"dcfb3888fbe298f87ff3bd8f72c7447b12e3dbbb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-07-24 02:05:22.000000000","message":"Patch Set 12:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/58/741158/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"ece120c14a6e63092ccddaea0cd494dcb2a2d9f5","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-07-24 02:10:52.000000000","message":"Patch Set 12:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/741158-12/Lenovo-iSCSI : SUCCESS in 51:26.85\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/741158-12/Lenovo-FC : SUCCESS in 51:11.47","accounts_in_message":[],"_revision_number":12},{"id":"755c58247523055c23b12b4f0f3398c128376b28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 02:50:48.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0514fb3f116c442b8b97ac63236e8670 : SUCCESS in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4972332eaf84421d85393ce281b6a118 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93da2b2921cc483fadcd494375ac0928 : SUCCESS in 18m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba64a5e7baa54c4c97e18411769c1f7c : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c58f3fee3140e6b27f6b532ce3c96d : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/623bd610b289421eb50645cd63e2ff31 : SUCCESS in 1h 16m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d67ce9b43da4814bd0f3da9bf83de06 : SUCCESS in 1h 18m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/008b5a8e3b9f49daab305293c137ac22 : SUCCESS in 8m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eda410a25ad9403dbf13fb573d545408 : SUCCESS in 1h 23m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1b89bc919aec4516b2c5268aed880491 : SUCCESS in 1h 00m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7a1e3a4785a406182e40e9ba28d5a0d : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":12},{"id":"f79896c3448128cec2d015b5dd2bf03b4d637803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 02:50:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"79da61817a3fa48e1a04a6d5ca2be53eecb415c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 02:52:33.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1eb78addf24ae9b91bd831e5040405 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14f54299e31a477daf2a5d2dab45a84e : SUCCESS in 56s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"517cb6448b613b888476f5da75ae4ad28f60f744","revisions":{"2fc6eae95c17d3978e7f9a8645ea9af8b374229b":{"kind":"REWORK","_number":1,"created":"2020-07-15 11:23:58.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/1"}}},"commit":{"parents":[{"commit":"79dea7ed608c50e0c48daca328e0ebcc927295a1","subject":"Merge \"Disable siblings for the cindelib functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79dea7ed608c50e0c48daca328e0ebcc927295a1"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-13 13:23:56.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2fc6eae95c17d3978e7f9a8645ea9af8b374229b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2fc6eae95c17d3978e7f9a8645ea9af8b374229b"}]},"branch":"refs/heads/master"},"38725da45a8b7323b6df6f9d625d9e8e77ddd452":{"kind":"REWORK","_number":2,"created":"2020-07-15 17:18:47.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/2"}}},"commit":{"parents":[{"commit":"cdfee5608b725e6be35d85f5b04cf51dc756cac5","subject":"Correct a typo of the hpe driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdfee5608b725e6be35d85f5b04cf51dc756cac5"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-15 17:17:21.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38725da45a8b7323b6df6f9d625d9e8e77ddd452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38725da45a8b7323b6df6f9d625d9e8e77ddd452"}]},"branch":"refs/heads/master"},"2ff207a59ced8ed927c4e74bb153287d33ffac0b":{"kind":"REWORK","_number":3,"created":"2020-07-15 17:42:23.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/3"}}},"commit":{"parents":[{"commit":"cdfee5608b725e6be35d85f5b04cf51dc756cac5","subject":"Correct a typo of the hpe driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdfee5608b725e6be35d85f5b04cf51dc756cac5"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-15 17:42:09.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ff207a59ced8ed927c4e74bb153287d33ffac0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ff207a59ced8ed927c4e74bb153287d33ffac0b"}]},"branch":"refs/heads/master"},"06aafab43acf42cc2bec8d95c85e1fafe94c174a":{"kind":"REWORK","_number":4,"created":"2020-07-16 12:31:47.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/4"}}},"commit":{"parents":[{"commit":"cdfee5608b725e6be35d85f5b04cf51dc756cac5","subject":"Correct a typo of the hpe driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdfee5608b725e6be35d85f5b04cf51dc756cac5"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-16 12:31:30.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06aafab43acf42cc2bec8d95c85e1fafe94c174a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06aafab43acf42cc2bec8d95c85e1fafe94c174a"}]},"branch":"refs/heads/master"},"fba5a0b10e4910c5af8391d0d87208254d5d3013":{"kind":"REWORK","_number":5,"created":"2020-07-16 20:32:31.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/5"}}},"commit":{"parents":[{"commit":"be4a682890e6c5aeee0f34891a80bfbe2aab7c6a","subject":"Merge \"zuul: collect cinderlib logs from tempest node(s) only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be4a682890e6c5aeee0f34891a80bfbe2aab7c6a"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-16 20:32:18.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fba5a0b10e4910c5af8391d0d87208254d5d3013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fba5a0b10e4910c5af8391d0d87208254d5d3013"}]},"branch":"refs/heads/master"},"d8831866936c00f37f54252e858bc9a04c7e0e15":{"kind":"REWORK","_number":6,"created":"2020-07-17 13:14:44.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/6"}}},"commit":{"parents":[{"commit":"be4a682890e6c5aeee0f34891a80bfbe2aab7c6a","subject":"Merge \"zuul: collect cinderlib logs from tempest node(s) only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be4a682890e6c5aeee0f34891a80bfbe2aab7c6a"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-17 12:25:36.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8831866936c00f37f54252e858bc9a04c7e0e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8831866936c00f37f54252e858bc9a04c7e0e15"}]},"branch":"refs/heads/master"},"0f43707ce33a9a03eefb073d51776106f82993f7":{"kind":"REWORK","_number":7,"created":"2020-07-20 09:43:02.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/7"}}},"commit":{"parents":[{"commit":"be4a682890e6c5aeee0f34891a80bfbe2aab7c6a","subject":"Merge \"zuul: collect cinderlib logs from tempest node(s) only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be4a682890e6c5aeee0f34891a80bfbe2aab7c6a"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-20 09:40:05.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\n* Supported Protocols\n - FC\n - iSCSI\n\n* Supported Features\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f43707ce33a9a03eefb073d51776106f82993f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f43707ce33a9a03eefb073d51776106f82993f7"}]},"branch":"refs/heads/master"},"23f51e7dc3f8fb3ff6cc56baee953bf70ad35eac":{"kind":"REWORK","_number":8,"created":"2020-07-20 13:54:17.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/8"}}},"commit":{"parents":[{"commit":"be4a682890e6c5aeee0f34891a80bfbe2aab7c6a","subject":"Merge \"zuul: collect cinderlib logs from tempest node(s) only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be4a682890e6c5aeee0f34891a80bfbe2aab7c6a"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-20 13:53:49.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\n* Supported Protocols\n - FC\n - iSCSI\n\n* Supported Features\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23f51e7dc3f8fb3ff6cc56baee953bf70ad35eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23f51e7dc3f8fb3ff6cc56baee953bf70ad35eac"}]},"branch":"refs/heads/master"},"0e9130846a3fbe2ddbf86b2b6ac8543e9fa1fb47":{"kind":"REWORK","_number":9,"created":"2020-07-21 11:49:20.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/9"}}},"commit":{"parents":[{"commit":"be4a682890e6c5aeee0f34891a80bfbe2aab7c6a","subject":"Merge \"zuul: collect cinderlib logs from tempest node(s) only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be4a682890e6c5aeee0f34891a80bfbe2aab7c6a"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-21 11:48:57.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\n* Supported Protocols\n - FC\n - iSCSI\n\n* Supported Features\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e9130846a3fbe2ddbf86b2b6ac8543e9fa1fb47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e9130846a3fbe2ddbf86b2b6ac8543e9fa1fb47"}]},"branch":"refs/heads/master"},"83cc2b4b08ea8304e609feee543fc59d7237605e":{"kind":"REWORK","_number":10,"created":"2020-07-22 11:26:32.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/10"}}},"commit":{"parents":[{"commit":"be4a682890e6c5aeee0f34891a80bfbe2aab7c6a","subject":"Merge \"zuul: collect cinderlib logs from tempest node(s) only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be4a682890e6c5aeee0f34891a80bfbe2aab7c6a"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-22 11:25:21.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\n* Supported Protocols\n - FC\n - iSCSI\n\n* Supported Features\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83cc2b4b08ea8304e609feee543fc59d7237605e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83cc2b4b08ea8304e609feee543fc59d7237605e"}]},"branch":"refs/heads/master"},"4e61c6beecc922280df6378f5d21591ddcef1e7d":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-07-22 11:27:50.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/11"}}},"commit":{"parents":[{"commit":"be4a682890e6c5aeee0f34891a80bfbe2aab7c6a","subject":"Merge \"zuul: collect cinderlib logs from tempest node(s) only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be4a682890e6c5aeee0f34891a80bfbe2aab7c6a"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-22 11:27:37.000000000","tz":0},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\n* Supported Protocols\n - FC\n - iSCSI\n\n* Supported Features\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Revert Volume to Snapshot\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e61c6beecc922280df6378f5d21591ddcef1e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e61c6beecc922280df6378f5d21591ddcef1e7d"}]},"branch":"refs/heads/master"},"517cb6448b613b888476f5da75ae4ad28f60f744":{"kind":"REWORK","_number":12,"created":"2020-07-22 19:40:57.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/58/741158/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/741158/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/741158/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/741158/12"}}},"commit":{"parents":[{"commit":"be4a682890e6c5aeee0f34891a80bfbe2aab7c6a","subject":"Merge \"zuul: collect cinderlib logs from tempest node(s) only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be4a682890e6c5aeee0f34891a80bfbe2aab7c6a"}]}],"author":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-02 15:00:54.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"Ivan.Pchelintsev@dell.com","date":"2020-07-22 19:40:09.000000000","tz":180},"subject":"Add Cinder driver for Dell EMC PowerStore","message":"Add Cinder driver for Dell EMC PowerStore\n\n* Supported Protocols\n - FC\n - iSCSI\n\n* Supported Features\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Revert Volume to Snapshot\n\nImplements: blueprint powerstore-cinder-driver\nChange-Id: Icef5b38ba39eec761c1cfa70e2a66bc28ddf4cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/517cb6448b613b888476f5da75ae4ad28f60f744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/517cb6448b613b888476f5da75ae4ad28f60f744"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
