)]}'
{"id":"openstack%2Fcinder~745415","triplet_id":"openstack%2Fcinder~master~I07a689c80aa08f98d1face108a9f379ddd7e8606","project":"openstack/cinder","branch":"master","topic":"cinderlib-func-bindep-prob","hashtags":[],"change_id":"I07a689c80aa08f98d1face108a9f379ddd7e8606","subject":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures","status":"ABANDONED","created":"2020-08-07 20:49:21.000000000","updated":"2020-08-13 22:22:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f5fdae8b73c16693cfe03632496ad74a46abf214","_number":745415,"virtual_id_number":745415,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"date":"2020-08-13 18:42:39.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"date":"2020-08-13 08:43:12.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"date":"2020-08-13 08:32:34.000000000","_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2020-08-13 17:09:05.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2020-08-13 14:25:00.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"date":"2020-08-13 10:13:17.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2020-08-13 08:55:18.000000000","_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-13 12:15:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-08-13 13:47:54.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"date":"2020-08-13 08:08:48.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"date":"2020-08-13 16:53:48.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"date":"2020-08-12 14:53:59.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"date":"2020-08-13 13:17:45.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"date":"2020-08-13 08:04:06.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"date":"2020-08-12 20:34:57.000000000","_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"date":"2020-08-13 16:10:53.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"date":"2020-08-13 09:40:09.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},{"date":"2020-08-12 14:10:00.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"date":"2020-08-13 08:20:40.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"date":"2020-08-13 14:24:18.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"date":"2020-08-13 12:44:43.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"date":"2020-08-13 14:22:25.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"date":"2020-08-13 12:34:57.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2020-08-13 09:15:56.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2020-08-13 00:37:31.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"date":"2020-08-13 15:04:26.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-08 01:54:06.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-08-08 07:24:04.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-11 03:58:24.000000000","updated_by":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"reviewer":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"state":"REVIEWER"},{"updated":"2020-08-12 14:10:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-12 14:53:59.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-12 20:34:57.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 00:37:31.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 08:04:06.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-08-13 08:08:48.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 08:20:40.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-08-13 08:32:34.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"REVIEWER"},{"updated":"2020-08-13 08:43:12.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-08-13 08:55:18.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 09:15:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 09:40:09.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 10:13:17.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 12:15:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 12:34:57.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 12:44:43.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 13:17:45.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 13:47:54.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-08-13 14:22:25.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-08-13 14:24:18.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-08-13 14:25:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 15:04:26.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 16:10:53.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-08-13 16:53:48.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-08-13 17:09:05.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 18:42:39.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58b5bc2b420a92b8de83b5b1b44407970965e57a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-07 20:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd4b53e9ea75203d7f15f56fd19b2093eb2fc3e0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-07 20:58:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/034a07b : FAILURE in 4m 23s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/0c0d456 : FAILURE in 5m 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"004de01c54d4a8dd2370b7110057d598df0bd46a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-07 21:01:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"450001cdea45fdeb69b1b4d98690b43a7c1e7ff7","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-07 21:06:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/be44374 : FAILURE in 4m 20s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/0bf8032 : FAILURE in 4m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"a1ee1009ce3df5b78f343e16a376858e06706826","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-07 21:40:38.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3dac738e66838c3cee75fe483a993204e29fc644","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-07 21:52:50.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e378b1412d642b3e387de2e507fe9d02e12aaf74","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-07 22:27:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/2/check/dsvm-tempest-cisco-zonemanager-job/7bbd20a/ : FAILURE in 33m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c590ab9694816e0deeac739f4b433cb905d3d4e3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-07 22:34:53.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c70d70c1c2e20d9102021c3f5d5bb0a440e5b0bc","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-07 22:37:32.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/2/iscsi/ : FAILURE in 1h 35m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/2/smb/ : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":2},{"id":"21a0769491ec05b155f5f5f36837ea38bd0b295e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-07 22:38:42.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"7d9dbe48706e9e95fd52ade765a210a63e464126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 22:40:34.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"ec7ab6fce0447e1e1960ccb858d3f05bc5d3022b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-07 22:41:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad018f7efe95acaf18ba21bdf04750d8471bd085","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-07 22:45:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/27ea9bc : FAILURE in 4m 24s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/a3e16a1 : FAILURE in 4m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"150fa2a32a874746f321e435126b98c7b602e3ba","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-07 22:54:44.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-15-745415-1/ns5_nfs : SUCCESS in 1h 59m 11s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-15-745415-1/ns5_iscsi : SUCCESS in 2h 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"20bec27ebd1f71cecb9a566ba9c7810c0357f503","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-07 23:03:26.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"25ae79afc6e27ed20dba5a9ea1ee014c25db17b7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-07 23:06:15.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-15-745415-2/ns5_nfs : SUCCESS in 2h 4m 16s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-15-745415-2/ns5_iscsi : SUCCESS in 1h 54m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5541898c46bae6fc82a6dbcb937b43802ce4803e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-07 23:21:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/2/check/dsvm-tempest-cisco-zonemanager-job/22f0576/ : FAILURE in 34m 34s","accounts_in_message":[],"_revision_number":2},{"id":"0f593b91b11751ed864437c5429f9e3bf831a2d5","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-07 23:32:19.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e872ecf913e2f5c0f9c9dbcba415f6fe68373ada","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-07 23:50:41.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/2/iscsi/ : FAILURE in 37s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/2/smb/ : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":2},{"id":"75e15366a192a7dd09329635b60348af7f1978b4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-08 00:22:07.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"6bbfc61a4fb428113b98efac66e95eb0fe4970b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 00:31:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36928534d14a45af9cbfc1145fa58ba4 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e31675ee994b5ba8a617d7d77a55c4 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/941d7a14c3cd472e815a79985bff8d5f : SUCCESS in 13m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc1e20b950fe4d52be2a2572de34444c : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/351fb076252e4ed8927c52d7601dea94 : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/66c2ccba6442415bbe09b90dcda8e1dc : SUCCESS in 56m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b486a8c2fb94cd19ed89846d8dae13a : SUCCESS in 1h 34m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/74bac590a7cc4e79aa913174d5102dfa : SUCCESS in 11m 21s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d385981dceb41009a8968e339e8a944 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bcc75fad98ba46e6806ddf20fb1a4273 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f8e19def3f9c4739b678595298c0b32c : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/22645d9326914817ab42e73dc4f383bc : SUCCESS in 1h 02m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a93bfbee746f4bfe86168c82cad7ab36 : SUCCESS in 4m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/22a385cf327b4b629f97e293b01b30f7 : SUCCESS in 1h 00m 24s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e893a7dfc24e45909fb5a5176b9711b9 : FAILURE in 1h 34m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b7d1608d14342a2b8ca4c7284fb692c : SUCCESS in 1h 15m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/85594ff8c3cc4ce8b4d804499c93676c : SUCCESS in 1h 41m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0e29f8c4bb914213931c291b1801a270 : SUCCESS in 1h 18m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cb615899238e41e99821874b6af11e07 : SUCCESS in 1h 47m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cfc67e84b9849e7859d16f9d38577af : SUCCESS in 1h 23m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2b03e745bc0c5be808b1feb67452291f9ea6d352","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-08 01:35:57.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"79e796f9a426646605d965449fdfdcb2128317b6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-08 01:54:06.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/15/745415/2/check/DellEMC_Unity_FC/b881e7a/DellEMC_Unity_FC/378 : SUCCESS in 1h 14m 02s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/15/745415/2/check/DellEMC_Unity_iSCSI/1ffd29b/DellEMC_Unity_iSCSI/367 : SUCCESS in 1h 38m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"63f80bc05186bbe6d1759402e2a930e0d5da60b5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-08 01:56:55.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"21586127176fc250d820a155f4ca40a64ad191d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-08 03:41:17.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7d1755ab96a651625c6c8ec1c65ac6634a96d24e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-08 05:13:05.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/tempest-dsvm-full-bionic/4b41f2b/ : SUCCESS in 1h 36m 03s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/devstack-plugin-ceph-dsvm-bionic/e699009/ : SUCCESS in 2h 06m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e5af9f0c5d43ac2d7081deefb0b3b5f323a798e4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-08 05:48:01.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"29f2bd99e304c403cffab6d1f650b9adcec5e0e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-08 06:28:11.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"421c717ed588869175b700f91e0e4c15494533b8","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-08 06:50:27.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/dca61b7cdfe3ee912e52dba644d4e482aaab907f/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"054e800e7ca6a0111fd76d9748744dbd8d88d65a","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-08 07:24:04.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/ba3e3bcbb9cf10df5168e4b6ac2bd9adf3dbdf09/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8abff2405c20563f4a95d95c78482d85c92004dd","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-08 11:28:44.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/745415/2/check/DellEMC_XtremIO_FC/5fce96a/DellEMC_XtremIO_FC/629 : SUCCESS in 1h 33m 53s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/745415/2/check/DellEMC_XtremIO_iSCSI/e5c8552/DellEMC_XtremIO_iSCSI/601 : SUCCESS in 1h 35m 49s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"e8aa76101e35d07e799246e7246f0c42e0ecc0e6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-08 12:04:22.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"238eabfc85d8a827d6de86f4686c35eed94a72f1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-08 12:44:52.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"37cf81e8dd81c47e48909edf731ff7bbc8f2ec6c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-08 20:35:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68529d51bcd5743665965095996ca77ee5f017bb","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-08 20:37:20.000000000","message":"Patch Set 2:\n\nThe patch seems to work: the cinderlib tests have been executed. The failure is a \"usual\" flaky one.","accounts_in_message":[],"_revision_number":2},{"id":"9ee230a8e2dc5099a0f61ef2c489429e7a889298","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-08 20:40:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/aa00fc1 : FAILURE in 4m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/e31b7f5 : FAILURE in 4m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"e84f531b7be7e84ec864d13e19063d64a3f091c0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-08 21:11:12.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"840bff1935e18abf459ab86795b508f2833e1900","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-08 21:11:54.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/745415/2/check-not-vote/ext-cinder-zuul/01b4856 : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c5a27f3f08e53337fd8b8e211321b3f431ff275a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-08 21:15:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/2/check/dsvm-tempest-cisco-zonemanager-job/6f39769/ : FAILURE in 33m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3b3c5df0b8531a865c1253bdc3969023d8e64ac3","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-08 21:27:08.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ea13db4de76e8c710241c31c73cacd3219968fdb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-08 21:34:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/2/iscsi/ : SUCCESS in 58m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/2/smb/ : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0c37821ac2e844220731ef13537ff375133aabd8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-08 21:56:14.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"fd22991aa5561c4d196e533572aa1346fbac6ff1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-08 22:11:43.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/tempest-dsvm-full-bionic/5a74835/ : SUCCESS in 1h 31m 36s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/devstack-plugin-ceph-dsvm-bionic/ab73131/ : SUCCESS in 1h 24m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5d890bb19bf84aa566f6530e5a562aa689e22953","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-08 22:17:20.000000000","message":"Patch Set 2:\n\nThe Zuul checks are going to fail \"thanks\" to a random failure in cinder-grenade-mn-sub-volbak, but cinder-tempest-plugin-lvm-lio-barbican is working.","accounts_in_message":[],"_revision_number":2},{"id":"a15c56b29bbe565482457cc97297714048d3d671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 22:42:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f4793bd19974102abded95f8ba3b2e8 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a3be897b8c424d876d00a071d8baf3 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9859e1040d2a49ccbaba0d38083619b7 : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f68b74196db34822baf94b4288298bca : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9f8180da59456682d34196686cb189 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/16387bf2e2034d23870408a25764d0c4 : SUCCESS in 59m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb647f6b0fd842599d81ecc680cb2405 : SUCCESS in 1h 05m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/35c38dd95b194d82b708a21d476e9f31 : SUCCESS in 10m 32s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/015977775f6842c191b05d2f91d9aeb8 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7e65017747d4243b34904066b387b4f : SUCCESS in 5m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41ce97d629974fa892cf59507f4b39a5 : SUCCESS in 7m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/601786fb37114d56a0cc2c75398c61da : SUCCESS in 38m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e520965c4c64df9a7bb0d3b7f6cb256 : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8412eacdc6534e48ba767a8e1e45ad5e : SUCCESS in 1h 04m 19s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3eb17d50495d45b9992db4b1e198c861 : SUCCESS in 1h 27m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd33995299d8413482f1a9486a1219cd : FAILURE in 1h 17m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/862f0179afe4478a8307c8640b01432f : SUCCESS in 1h 28m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3c5b90a00d0e48f781dea3f50201f916 : SUCCESS in 1h 11m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dde2b12dee38494a86f217bcd39a59f6 : SUCCESS in 2h 01m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2ff891d88784e5cac4fc04c460d0ea6 : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d02ec18b6dead6ef3c181d0af0cb8453b2e35f94","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-08 22:46:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1d2baa3adf4b6f6da6ec7180b3fdf8e51a66bd6d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-08 22:47:03.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4abdd9cb49ec4df49b53c12d3192f02a214c4d72","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-08 22:51:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/5409200 : FAILURE in 4m 46s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/965c9e6 : FAILURE in 4m 42s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"5f177cd51c48f5f74d8c53047091858e96b1ad32","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-08 23:26:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/2/check/dsvm-tempest-cisco-zonemanager-job/5a02049/ : FAILURE in 33m 56s","accounts_in_message":[],"_revision_number":2},{"id":"58ef701362517249d86d653e89b356d0b78f61ea","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-08 23:37:56.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"db750e98ed51f5c523a922af4f17295fa56c19b7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-09 00:15:32.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/2/iscsi/ : FAILURE in 1h 28m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/2/smb/ : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":2},{"id":"08a6f68714244e4b18ed0b25902e5ac810dfa79e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-09 00:19:21.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/tempest-dsvm-full-bionic/46f48f6/ : SUCCESS in 1h 29m 09s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/devstack-plugin-ceph-dsvm-bionic/3cdd4f5/ : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b6c547e21194e5fb68cb1e4fb1b2aebf61be9e35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 00:38:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4ff25b3bb674bafbdddc654aefa42d3 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/893e0c046bfb4260a0428b108906531e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2527c709bf25462f90310f0a479c2856 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf3597fe8828410dbb11a0786a89cdce : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8555d2d2d14f67b5b71d2439c2ec35 : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/fd9cca582d404143a626d008ecb48b7f : SUCCESS in 54m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3152e644979e4fda821cedc7dec01058 : SUCCESS in 1h 19m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/437e23e38fbe4eedabdac5a77ab4669f : SUCCESS in 11m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c13a0f170ac34863a9a54a275afc86f1 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cc0760b7a1c4b4092f53f44caf37110 : SUCCESS in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d089d37fbfa435b8a9c65e92ad2c478 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc804d40d39741e59cb79d4cd0e9c283 : SUCCESS in 43m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f842f5a9b134184bc99389606f55b5b : SUCCESS in 4m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/abc5b7ca4a934acf852b19fecf8b6d2b : SUCCESS in 1h 13m 33s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/44f47ff13e004f51acc190d8649e36e4 : FAILURE in 1h 34m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2ab060f747884f82ba4003f20588396f : FAILURE in 1h 22m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d85f06375b4b4bdcb3198c5a23d7b765 : SUCCESS in 1h 35m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4443fbf2a8d04262a251448141935469 : SUCCESS in 1h 25m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/06a2575ef7cd486e9e91c9a1d2f2debb : SUCCESS in 1h 48m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f1e02f4decb47cbbed3864429f8449f : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4cb4de6c8d3fb778cfe5bd0fcc6f45a7b56a0698","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-09 00:49:04.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"86c9a79c3befccc624deb7dd025e2c8b42c7d1b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-09 00:56:27.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8336d7be76c8702357016511d9bd28e46e8b5f37","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-09 02:33:12.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/745415/2/check-not-vote/ext-cinder-zuul/4993754 : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b85477f958d0afc098d6e02f93a31a42178c49a6","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-08-09 05:20:42.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/15/745415/2/check/DellEMC_PowerMAX_FC/79e6aa9/DellEMC_PowerMAX_FC/429 : SUCCESS in 3h 36m 06s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/15/745415/2/check/DellEMC_PowerMAX_iSCSI/d50ac51/DellEMC_PowerMAX_iSCSI/361 : SUCCESS in 4h 56m 07s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"3e78e855cd0a624ed43bc9b60d97f88d5f33aa23","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-09 05:54:58.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a65bb29ab3c6e1dacea26e4814d04f03e9c3d870","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-09 06:39:11.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"d9ceca07c71ba92285688fbe4b548d872e361c9e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-09 11:15:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68ba7b2015a6354ceb20bfc7bb29a8cfadd5cf85","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-09 11:19:38.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/bb5d9fd : FAILURE in 4m 20s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/d77a46b : FAILURE in 4m 18s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"51923f6fa03be4de1a3c692e5e071c5777b3814d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-09 11:32:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c5759828b272335406a331093bc94812d68a094f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-09 11:48:45.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a4c45d836d0ea089551cf7056f7b441939bf067d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-09 11:54:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/2/check/dsvm-tempest-cisco-zonemanager-job/b07e620/ : FAILURE in 33m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4776a615edd2da36551ad3124502c965a4d36d3a","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-09 12:06:23.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"94dd64950307e6f5d2a023703634321e2eeba959","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-09 12:11:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/2/iscsi/ : SUCCESS in 56m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/2/smb/ : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b495977cc01a2f944d3d5d05b1b6df909ea9b4e7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-09 12:47:44.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/745415/2/check-not-vote/ext-cinder-zuul/8684b1e : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":2},{"id":"fe27b3bdfcb5a360822c2823373df8a11f328d98","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-09 12:58:19.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/tempest-dsvm-full-bionic/3d96ebf/ : SUCCESS in 1h 28m 49s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/devstack-plugin-ceph-dsvm-bionic/ffd755a/ : SUCCESS in 1h 27m 42s","accounts_in_message":[],"_revision_number":2},{"id":"8fcb5b72060bb7526be56b82cc91d5507554a1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 13:05:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d426e72050946a19a6809752977c708 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2f0d1b321449128aa70ef02344a611 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a82d78649f3947f599db9af6abe715f8 : SUCCESS in 12m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7732d5feef7463eb110db1d6d72b922 : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2002ebff14a4e1ab304462b72ead979 : SUCCESS in 17m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/7c726a534dad4d2396ee4c9ca5913626 : SUCCESS in 1h 09m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/036964bcd6df4a8c8d6ad901d701212d : SUCCESS in 1h 01m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2bb1b8f181fb418eb9a4c377fcfffa4b : SUCCESS in 11m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/547cae3d172941b79a081650f2bb888e : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8577b5e30c943a9898dc579a8147f35 : SUCCESS in 5m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/efee5f7e097f4bb4a9f3a858b57444f3 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ad94d570c40941429c42cb0782f3ac25 : SUCCESS in 38m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/84fc7c5d024248909f2a1944fdaf2983 : SUCCESS in 7m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a7cf77c957c4dbca6a8ace4605869d6 : SUCCESS in 1h 04m 18s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/61fb5bb76e6043d7abc839d18e51f5d5 : SUCCESS in 1h 37m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/09aad3667ee44490a80dd49444e13d61 : FAILURE in 1h 10m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/12093982921b439e95b44191164910f0 : SUCCESS in 1h 28m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1b661ead5f34d07a9216561b1842e43 : SUCCESS in 1h 29m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a64080e8398b48b19a9d7f543996b6f4 : SUCCESS in 1h 48m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1dcb9d468331441295a7b3aaef6e1cf6 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":2},{"id":"cb5631272c3cadd9238eb431b582423a448b8303","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-09 13:24:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f13efd2a17a4e37819585ae57e79cafa82246dfd","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-09 13:29:00.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/32855bc : FAILURE in 4m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6ce078b : FAILURE in 4m 30s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"67912eb9565858aace3f4177dfae731a92583116","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-09 13:31:32.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"b51c005ce84a01d6bffd3ef0f13cf56c6fc60504","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-09 14:03:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/2/check/dsvm-tempest-cisco-zonemanager-job/2e8c230/ : FAILURE in 33m 37s","accounts_in_message":[],"_revision_number":2},{"id":"0b46204436c703685f79034d1a4a925cee140180","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-09 14:06:02.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/745415/2/check-not-vote/ext-cinder-zuul/4e62a3c : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":2},{"id":"4caaffc32818c637de352e43d95e2f53acf04e23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-09 14:12:00.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"df05afb0f30aef4985472fd9b538baada8fccf70","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-09 14:15:37.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"b681dd59c9c6dcb754e040fbaad67e114805a3a1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-09 14:19:00.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/2/iscsi/ : FAILURE in 41m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/2/smb/ : SUCCESS in 54m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e21f910df013293c8834745a583b26eca4b83e5a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-09 14:55:04.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"fa8f3f37b9e1190521a75dc9adac54d29914ab7f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-09 14:58:31.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/tempest-dsvm-full-bionic/ed365b1/ : SUCCESS in 1h 29m 11s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/devstack-plugin-ceph-dsvm-bionic/1dcbe6b/ : SUCCESS in 1h 27m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c82f5fa1dd569a80dc538a984cbb3f41ef2c2f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 15:15:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61bc105daa2d429fb34725b5b5b58972 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e055db765d4849aac1a087e7c332bd : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0248d8fd09394b6580bef1c33d45a561 : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f638eabcb3f4eae9a11e896d98f8012 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb441d73995147978f12d0cfc0ad319e : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/dffad0c19aa8418d82070727338ced3f : SUCCESS in 1h 06m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c490d27af164fcdb7e04b1cd9b87e26 : SUCCESS in 47m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/beb25c69336d45c1b2aa43f4a0987025 : SUCCESS in 8m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cde265864c284d7387399aa14c8632ac : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98503edf8f334c60bdc69e4676868454 : SUCCESS in 6m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64f55bbf03ce40adaad1436a8ae2863b : SUCCESS in 6m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/17ccc3c7ac7e43d282e3b2e4593c55d3 : SUCCESS in 32m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/55198c6f2fe34df58a06666a3dc03433 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/02fc2ebe84244878a76d2e7fb5fbf315 : SUCCESS in 1h 00m 02s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/25d4784b617c4c34ac15614cc3f9a893 : SUCCESS in 1h 30m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8aa009666cba4bfdb26824139082e2b9 : FAILURE in 1h 08m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c05b775f66884225ae8f65d8cc798efe : SUCCESS in 1h 16m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6cab1ffe9715405db230d83d7b42cc99 : SUCCESS in 1h 17m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2cc8cd38c6d4d2b99728c6ce5ceaaef : SUCCESS in 1h 47m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d8dd47e1516464598c00e42e0791bed : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":2},{"id":"51af4098c76cc8026ffe11f0cd12a57aa813f660","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-09 15:32:41.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1f029c276660e8fb40225e4cb6eb4da6fdb1c6c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-09 15:36:10.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"3506d3fc04f0f2f8543d15d8ba4ab7e59b3d0a10","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-09 17:56:19.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"49c5c333361b7aee02db5e29aa933467c4cbe14a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 20:19:18.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- tempest-cinder-v2-api https://zuul.opendev.org/t/openstack/build/5a4292052c9b4bab849542323702bb61 : SUCCESS in 52m 10s\n- legacy-tempest-dsvm-multibackend-matrix https://zuul.opendev.org/t/openstack/build/3d0a688b74ad46c7a7eb981c11c9a534 : FAILURE in 52m 10s\n- cinder-grenade-mn-sub-volschbak https://zuul.opendev.org/t/openstack/build/0e9f63218e534b95bad7ca154241aa19 : FAILURE in 1h 20m 23s\n- cinder-grenade-mn-sub-bak https://zuul.opendev.org/t/openstack/build/e249d272899c49bfa4ba4438f3b369d1 : FAILURE in 2h 11m 53s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d50ceda9d9de4ff19fe078b1d17473c6 : SUCCESS in 43m 12s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/6c7ff65c211240abadb6fd858b69a343 : SUCCESS in 54m 41s\n- tempest-full-py3-opensuse15 https://zuul.opendev.org/t/openstack/build/0509ba7275e74830accafe0eb6c692ce : FAILURE in 52m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea1f16e2f85f179bc9f29d77499c5a008637f5f2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-09 20:28:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"19909104e36c4104bd65bd9c2a7f2aedbda8705b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-09 20:31:06.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"065f5ff722d646dd106a21ac8c6ed2026d98b4ed","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-09 20:33:37.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/00e7c28 : FAILURE in 4m 18s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/4e4ebe1 : FAILURE in 4m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"38ed4c824e33c2c50653788ef66955412861771f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-09 21:04:22.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/745415/2/check-not-vote/ext-cinder-zuul/3b13264 : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":2},{"id":"2adaad88114470daa162719886367fbad4226fce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-09 21:07:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/2/check/dsvm-tempest-cisco-zonemanager-job/9926f24/ : FAILURE in 33m 43s","accounts_in_message":[],"_revision_number":2},{"id":"bf3e5eb2349c1ff308fda1a921c634749cd57993","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-09 21:19:38.000000000","message":"Patch Set 2:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"61dedc5d1b4e40f271ff7eaf4834a2e474470343","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-09 21:28:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/2/iscsi/ : SUCCESS in 59m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/2/smb/ : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":2},{"id":"aef085322eedfb2e82a2f259b696d4ad7aa12c4b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-09 21:47:50.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4ece64888c9746c4ffc82006ae8410aba7d82656","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-09 22:02:10.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/tempest-dsvm-full-bionic/ae391ff/ : SUCCESS in 1h 30m 31s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/2/check/devstack-plugin-ceph-dsvm-bionic/7c6ebfd/ : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c2d55f3299d7839149a351a07b0e29d38d00f4a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 22:18:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91a632ee56d64174a98d1d32e123ebc3 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab22637110ee4af08868f6097c66ee91 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6717d59250746899fce2e504f759044 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59062342ccbd4d9aadb40753eea2a47a : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d44721b8f7b4af1b1226708e203ed12 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/8966440b52394bf2b7173225e6214c0e : SUCCESS in 1h 03m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8e6a54597a445babd1cf2ec4f761ef3 : SUCCESS in 1h 03m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e875b447261644ddaff56228e9222f31 : SUCCESS in 10m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8964700de7d84329be994c42939a7fc4 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/119baa87c2ea4eef903a80588779fe3f : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/549eb67c50394f74a495f86325c7ae6d : SUCCESS in 4m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/554d6fce6be049e4b8c5785c45f82206 : SUCCESS in 43m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22252dbcad954d40a4d3d6fd1dbddb21 : SUCCESS in 4m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f668f283d5684082b48bccbf40bec85a : SUCCESS in 1h 22m 44s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f13a5f46308a454c9c6e4d5d66178cbf : SUCCESS in 1h 44m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d536f0d3de7048c9affa7f2370d78df1 : FAILURE in 1h 07m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9be293d720e2488588ab94f026c711f2 : SUCCESS in 1h 31m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e6765815f1ff4cabb684826873de6ef8 : SUCCESS in 1h 27m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26bf8e6547004aae95325207b105fdc9 : SUCCESS in 1h 43m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e63ba17e8a1e495fad054185a3424ca6 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":2},{"id":"76a0419a757354c51ad7ed5a23b8ebc1851f3ed6","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-09 22:37:31.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2050cf75523b8bb12b93168aa8f9f9e9acf323aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-10 00:41:51.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8cc1d36fff2d6b563e18abd35c704bb349468192","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-10 16:27:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/15/745415/2/check/nimble-iscsi-driver-dsvm-volume/5f4c40f : SUCCESS in 2h 33m 58s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/15/745415/2/check/nimble-fc-driver-dsvm-volume/8933b3d : SUCCESS in 2h 20m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3d5889ff8f0796d87b838360c331973e1a2da928","author":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"date":"2020-08-11 03:58:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e95fa572258a47ef2bc33c8f01294db4a294f94d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-11 04:03:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/00c6f2f : FAILURE in 4m 28s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/7ef460a : FAILURE in 5m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"b88baf1a635b3fd11b674df9fc1e71fe3e99173d","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-11 04:49:55.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cbcd9b9c333e0fdd8d8906aefa130758a7153d94","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-11 05:31:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/1/iscsi/ : FAILURE in 1h 32m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/1/smb/ : FAILURE in 26m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9af75f3944789335772f8aa7674b76b576281b9f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-11 05:43:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/1/check/tempest-dsvm-full-bionic/15993b5/ : SUCCESS in 1h 30m 13s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/1/check/devstack-plugin-ceph-dsvm-bionic/94e3ed6/ : SUCCESS in 1h 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"db9ed6f341633522098d06319fad1eeb63921740","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-11 05:45:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/1/check/dsvm-tempest-cisco-zonemanager-job/3b714f1/ : FAILURE in 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7383237bc6d4384c6c5d9effc6f096f52e650c5f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-11 06:04:49.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"84ab2f7ab667fd2a75f5a687d6fbba1ebe94cd45","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-11 11:29:49.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2d90c5fbad25fc3e9aa8c0e7091f57ab35587ea5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-12 12:49:26.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"592b2f75820d3bff29777320df03fe2b459e2aad","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-12 12:49:48.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"df72846170210c1eba4607035d1ba117af6a4999","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-12 12:51:22.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"371c1ca43ffa0b5d604b74fdd4e5e9b1a9ab80bf","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-12 12:54:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/287967a : FAILURE in 4m 36s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/dfd8cf4 : FAILURE in 4m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"f350d194daf56357cc4352564589d100d7f6933d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-12 13:24:43.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"5155f9cbfc2b5455e3b7274b85ae7b15a4a4e648","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-12 13:29:10.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/745415/4/check-not-vote/ext-cinder-zuul/7546746 : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":4},{"id":"7b3a388eee849353d5753a3c6734de7632ce0bfb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-12 13:29:19.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/4/check/dsvm-tempest-cisco-zonemanager-job/606eb82/ : FAILURE in 32m 59s","accounts_in_message":[],"_revision_number":4},{"id":"5cf362e6e10adb43875b76caff338881bfaa1793","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-12 13:37:21.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"0e81491e4d13c16ed9f3731669ac6066bb43a58e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-12 13:40:41.000000000","message":"Patch Set 3:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"a51d16aa6f45da66820c6d64f41d54ba5e7700b7","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-12 13:40:59.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"71db1d7a11f81f5a0a00ad4b39ef7eef3ece018a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-12 13:47:41.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/4/iscsi/ : SUCCESS in 57m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/4/smb/ : FAILURE in 41m 23s","accounts_in_message":[],"_revision_number":4},{"id":"33306a098659619a9848cf24e80f1a0ff01f20a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-12 13:48:45.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"006c47450845659695175fdc9d1878d86506ff61","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-12 13:57:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/745415/4/check/dsvm-tempest-zadara-driver/2e8cf5d : SUCCESS in 1h 07m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"02fd138ecce83cca4c22f52c01a365e126f3feed","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-12 14:10:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/15/745415/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e15590a/e15590a.html : SUCCESS in 1h 08m 28s","accounts_in_message":[],"_revision_number":4},{"id":"71c82d622ed18d2ab6fae23006862ff0c5ebd3e4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-12 14:24:02.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a90d3f752d0b543eac75244e9724fa0390f3e70d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-12 14:26:06.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/745415/4/check/DellEMC_XtremIO_FC/caa039e/DellEMC_XtremIO_FC/659 : FAILURE in 7m 46s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/745415/4/check/DellEMC_XtremIO_iSCSI/ac305df/DellEMC_XtremIO_iSCSI/630 : SUCCESS in 1h 35m 45s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"4563ef47a22892fce53d3b5d2f2f01b7aaf4d2b9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-12 14:30:37.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/4/check/tempest-dsvm-full-bionic/8030e1a/ : SUCCESS in 1h 24m 01s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/4/check/devstack-plugin-ceph-dsvm-bionic/fe028c6/ : SUCCESS in 1h 34m 49s","accounts_in_message":[],"_revision_number":4},{"id":"a05b3efbf01a815cbb75fa951fe322e95acf54c2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-12 14:53:59.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-15-745415-4/ns5_nfs : SUCCESS in 1h 50m 6s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-15-745415-4/ns5_iscsi : SUCCESS in 2h 3m 51s","accounts_in_message":[],"_revision_number":4},{"id":"929d0edaec37b2478ae5ba3f5db416462beea412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 14:58:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48041ba90c50436da4e5f0826c2b2970 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5a5e4147b44364ac97ce687b9b3ef7 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74b7598bf7c94fb49231f01ae205c034 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84afbca333e2446c8faadbd8b3ef12c6 : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59dbf838476f41d5a5a149116cbbd785 : SUCCESS in 15m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/04ff0f7e894a48ba987258ff5b7a542e : SUCCESS in 1h 06m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2857ffded20470d9dd661182a5035c3 : SUCCESS in 1h 08m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f05a891dbdf4a04a343ba669c48a4e0 : SUCCESS in 13m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/70a964a45c564328824b03add5d7bbdc : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2c0099e564c4711aa3abbc12d9faa57 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73de29ff0de649ae963d79de4017a533 : SUCCESS in 8m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3290ae296e524cf3ae66c4a3ecb83100 : SUCCESS in 44m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e52789c7ce5e470489f90744e0fe42fa : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/269a58df3d234a4a95cabd3d8a8856d5 : POST_FAILURE in 15m 12s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a6db92bd83bb447a9bde31a4d7da0ff7 : SUCCESS in 1h 35m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/62496ad1f15343329c4e52c9932ec668 : SUCCESS in 1h 08m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/67ea5f1618d74a2f890e1980c70a3c73 : SUCCESS in 1h 42m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5498a50954ff4b2fa2a52b58fb2fcfab : SUCCESS in 1h 33m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1f65907dcd9c4d8bb99c83f5d9e1c649 : SUCCESS in 1h 58m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82f08cf408844d49a2d924b150702d36 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":4},{"id":"863d90c94ba88e8ac4ee0256086dfe8eb9f649bc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-12 15:03:41.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-15-745415-3/ns5_nfs : SUCCESS in 2h 13m 50s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-15-745415-3/ns5_iscsi : SUCCESS in 2h 8m 58s","accounts_in_message":[],"_revision_number":3},{"id":"a9114a9ef2f53a1951c1ed2a269fb7a6322a5618","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-12 15:05:14.000000000","message":"Patch Set 4:\n\nrecheck\n\nA glitch in cinder-plugin-ceph-tempest. It\u0027s a glitch looking at the past history of the job.\n\nBut more important, both affected jobs (cinder-tempest-plugin-lvm-lio-barbican and cinder-grenade-mn-sub-volbak) worked","accounts_in_message":[],"_revision_number":4},{"id":"99666e4c176d5b838443f6bd450516c706413805","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-12 15:16:36.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"96cca4b088fe4d341a8ed3273ebed84124fe9853","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-12 15:22:49.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"0b6ab46b9141db0c0220ee21a0bc758de6eed73e","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-12 15:56:17.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"bfdaaba67f2173c9c23f31e28d39bd958f6d293f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-12 16:24:33.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/745415/4/check-not-vote/ext-cinder-zuul/42191a7 : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":4},{"id":"c04908721f013007b7c407dbfaf06b9fd40761ee","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-12 16:30:22.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/4/iscsi/ : FAILURE in 1h 24m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/4/smb/ : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":4},{"id":"d14f284157946d3192dad55ead0e381670623e4b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-12 16:46:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/4/check/dsvm-tempest-cisco-zonemanager-job/996e233/ : FAILURE in 33m 20s","accounts_in_message":[],"_revision_number":4},{"id":"116013939a540308d81262cfad8338863343140d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-12 17:10:24.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/4/check/tempest-dsvm-full-bionic/f22db15/ : SUCCESS in 1h 22m 13s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/4/check/devstack-plugin-ceph-dsvm-bionic/23a6994/ : SUCCESS in 1h 50m 40s","accounts_in_message":[],"_revision_number":4},{"id":"5ad73c5b9361752d906d6e9fa828862919404214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 17:28:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/623d8d6bed8346b5955ae36a20c7501d : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aadca4e218494117a6efa9abd976409f : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aea0007844394609923ba28580645a6f : SUCCESS in 11m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ca952ac4a5c45e0b77c9a7e43ee042e : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c4c1a8afa74ef0bbb10b0aa4e95ac9 : SUCCESS in 15m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1165c50f64bc4cdb80bfe0283e5d66ee : FAILURE in 39m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2344330ecdbe419b8bf9e435b9c3581f : SUCCESS in 1h 04m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/410b49e8ed084e1ab664706bdf37d475 : SUCCESS in 11m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2ecb34c194814103ae6e7b81ae28d70f : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81c2b75e30494097be24f311025e6613 : SUCCESS in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e5ba2f1d12741b79a1e8c06c34d25fe : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a821a762d8ab4364b6a3a8c90ddc1bc7 : SUCCESS in 43m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ddc5b82debe8428aaaae414150bc093f : SUCCESS in 8m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dd8f6346d05845ca9893fe8b1b9bdb1d : SUCCESS in 1h 24m 26s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d1cbc8ea9024be39b397f8089c606bb : SUCCESS in 1h 22m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f06fc2fd2674774be191830bebbf340 : SUCCESS in 58m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a70011efb4904e5284d6b3f31a93f94a : SUCCESS in 1h 29m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c8b2f7fbe1ee41deb712507fda8579f8 : SUCCESS in 1h 28m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f77c302a11a9417c977cc0586e830478 : SUCCESS in 2h 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9784c6bc46d34a02b7e392d0ff626ba5 : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":4},{"id":"125f92a3b365165a832d1b77435e3426d8c9dbe4","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-12 17:54:08.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"1ebce7227ab7f5aac3c9132d22c79fea9f7a6505","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-12 18:31:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/745415/4/check/dsvm-tempest-zadara-driver/9ea85a1 : FAILURE in 1h 09m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bde162f5ab4719b96202f7b1b8dd4c198735db38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-12 19:49:57.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"101dd3c0d3badb08edc709d52612d6b19917d7c3","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-08-12 20:34:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/15/745415/4/check/nimble-iscsi-driver-dsvm-volume/94d0a3b : FAILURE in 2m 09s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/15/745415/4/check/nimble-fc-driver-dsvm-volume/ab44af5 : ABORTED in 20m 51s","accounts_in_message":[],"_revision_number":4},{"id":"43a277864b02aea6adc425c61aea1745d003a376","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-08-13 00:37:31.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/15/745415/4/check/DellEMC_PowerMAX_FC/be6c3c1/DellEMC_PowerMAX_FC/450 : SUCCESS in 4h 16m 02s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/15/745415/4/check/DellEMC_PowerMAX_iSCSI/871309f/DellEMC_PowerMAX_iSCSI/382 : SUCCESS in 4h 27m 13s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"52519f2fb0294e910e07035610646fd151ab18eb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 01:12:53.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"2c1bc3127838a8827c3cc91dce9d121140b45ec8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-13 01:53:21.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c0c232292c6ff589208c07b857d3ae6d827d2d53","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-13 08:04:01.000000000","message":"Patch Set 4:\n\nI made https://review.opendev.org/#/c/745838/ a bit more generic and cleaner (hopefully)","accounts_in_message":[],"_revision_number":4},{"id":"ef10c9f775dce91769b86c96ddc5ac040b816ef6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-13 08:04:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bda7852e328378c95236621ff5f6d3aa3a0c5c51","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-13 08:08:48.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/15/745415/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/948e363 : FAILURE in 4m 24s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/15/745415/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/a1c74ac : FAILURE in 4m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"7f61e5743b0a1018a0c755cea9ae04a524ad52b1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-13 08:20:40.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/745415/4/iscsi/ : FAILURE in 16m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/745415/4/smb/ : FAILURE in 16m 09s","accounts_in_message":[],"_revision_number":4},{"id":"7db0592d17477e46ae15b0c088a17f9b9294956d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 08:21:45.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"631db8e490aefa64cb31d8f12fb154d6f5a99042","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-08-13 08:32:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1441/index.html : SUCCESS in 1h 23m 12s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1589/index.html : SUCCESS in 1h 42m 19s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1614/index.html : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":4},{"id":"1fcb8a4b44a7938f16ff0466c8d59110ac057e89","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-13 08:43:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/15/745415/4/check/dsvm-tempest-cisco-zonemanager-job/71b05bc/ : FAILURE in 33m 12s","accounts_in_message":[],"_revision_number":4},{"id":"d9fd519ebf2ceae59aa6c5e1b85463bc1e34814a","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-08-13 08:55:18.000000000","message":"Patch Set 4:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"71d2abb3f4fe90c6019f46aedb4440ff6ed715fe","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-13 09:15:56.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/15/745415/4/check-not-vote/ext-cinder-zuul/a4635bd : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":4},{"id":"d2fdaf686022b2b97e262ecad726c805f5dde078","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-13 09:40:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/15/745415/4/check/dsvm-tempest-zadara-driver/2ff8dd6 : SUCCESS in 1h 35m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2941f70bddf606e4c961bd75ae896b300c1724a3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-13 10:13:17.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/4/check/tempest-dsvm-full-bionic/a26b169/ : SUCCESS in 1h 35m 33s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/745415/4/check/devstack-plugin-ceph-dsvm-bionic/78953dc/ : SUCCESS in 1h 42m 03s","accounts_in_message":[],"_revision_number":4},{"id":"7a199a92976169b9a3dbd3f9f7358666e62db058","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-13 10:37:28.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"9437116a0f0a3df3ed24798e82d2ff104c91d0e7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 12:02:36.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"fbc86c746dc5e3c2f0065e2934c30902722960c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:15:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd18f901dce240c880764c4640193df4 : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cca44e29787481aad6afef9d45882a0 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/211c3d069b2549c3a9fc7a5cc2de22d9 : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe81268362b24aa5a94804e65a102a58 : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4142b09aa0fe42aabf787a81a262ae1b : SUCCESS in 14m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/6528acf09b394fa590f959341e06dd1c : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d65e377fd5da4fa4afdabee3395e8c48 : SUCCESS in 1h 03m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d408017f42e4ca2b4b22b391eaf1625 : SUCCESS in 13m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/494c6368d0834cabb669e5a78a0c95cf : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c73b927438c248cba9f1d75f9298bd4f : SUCCESS in 8m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc09f4f7cb894c73a993b86f76dcab04 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5df5caf10dd149e8bbbdee24fb92bdff : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1b969a0ff0d4450685efbe9ffe66d6b5 : SUCCESS in 4m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a12854cbf20435ab029ed3ba4280641 : SUCCESS in 1h 39m 44s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e89c546e2b704d6fbf923988686f9676 : SUCCESS in 1h 18m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e81633230b50456ab30b4db856b75c2e : SUCCESS in 1h 17m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/76270e0112e94547bcdc42c93a7cec6d : SUCCESS in 59m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/18e9b821869d4d95b41e04abd1862752 : SUCCESS in 1h 18m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/33fa53c6193d4f5ba26b6d5921d39e4c : SUCCESS in 2h 09m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaafc5b3623a49fcb6237f7cc331394c : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":4},{"id":"cd7295e14248699e9b8e41369fe61dcb7f9b8559","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-13 12:34:57.000000000","message":"Patch Set 4:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-745415-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"262c4f3c9c282f23b6862536e32080c2c5f1367b","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-08-13 12:44:43.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/15/745415/4/check/synology-dsm-cinder-iscsi/3cc0b72 : FAILURE in 28m 46s","accounts_in_message":[],"_revision_number":4},{"id":"41e3080b8cbb362c878b885ae586c6aa9ad1cec1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-13 13:17:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/15/745415/4/upstream-check/cinder-cDOT-iSCSI/aa777bd/ : SUCCESS in 34m 48s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/15/745415/4/upstream-check/cinder-cDOT-iSCSI-IPv6/32bf1fc/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/15/745415/4/upstream-check/cinder-cDOT-NFS/91617fd/ : SUCCESS in 35m 33s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/15/745415/4/upstream-check/cinder-cDOT-NFS-IPv6/5c5f07e/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/15/745415/4/upstream-check/cinder-cDOT-FCP/3fba6d2/ : SUCCESS in 1h 00m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"ea2a65d9cee80d4648f08fb28c8f4197eb569ceb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-08-13 13:47:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/15/745415/4/check/DellEMC_SC_FC/a65a571/DellEMC_SC_FC/292 : SUCCESS in 1h 08m 59s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/15/745415/4/check/DellEMC_SC_iSCSI/60fabf6/DellEMC_SC_iSCSI/266 : FAILURE in 1h 32m 07s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"6f96b8e0d5b81442e09e6cb429ebb361b41b4ff0","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-08-13 14:22:25.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/745415-4/Lenovo-iSCSI : SUCCESS in 48:55.65\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/745415-4/Lenovo-FC : SUCCESS in 45:23.39","accounts_in_message":[],"_revision_number":4},{"id":"a9587833432e98a5f75ee2e5163375c02dcb7108","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-08-13 14:24:18.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/745415-4/HPMSA-iSCSI : SUCCESS in 48:38.96\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/745415-4/HPMSA-FC : SUCCESS in 46:59.29","accounts_in_message":[],"_revision_number":4},{"id":"08c19238e1b8ac68b7977e3ae9183f1d6b2a95c2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-08-13 14:25:00.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/15/745415/4/check/DellEMC_PowerFlex/729f51d/DellEMC_PowerFlex/367 : SUCCESS in 2h 09m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"cfb0cae3d55b67b801ceabeffd695f724ea0b1f0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 15:03:41.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"36dc95040b47c5ae29980995d9cfd66222c0db5f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-13 15:04:26.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/15/745415/4/check/DellEMC_VNX_iSCSI/75eb746/DellEMC_VNX_iSCSI/261 : SUCCESS in 2h 48m 48s\n- DellEMC_VNX_FC http://publiclogs.emc.com/15/745415/4/check/DellEMC_VNX_FC/01f40af/DellEMC_VNX_FC/258 : SUCCESS in 1h 11m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"580de9eed9e5095e1f343b1acf1291180b588680","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-13 16:10:53.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/15/745415/4/upstream-solidfire-check/cinder-solidfire-iSCSI/50e4904/ : SUCCESS in 1h 02m 10s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"45dcfebda74c99ff696dddd4dccd55c3d15ae7ad","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-13 16:53:48.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/15/745415/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3878f4a1569abf9dc6198005c1d3ae70e1a3d12b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-13 17:09:05.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"5682ab4711be49da3ec96c28262d374ad0a468de","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-13 18:42:39.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/15/745415/4/check/DellEMC_XtremIO_FC/a44151f/DellEMC_XtremIO_FC/670 : SUCCESS in 1h 34m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/15/745415/4/check/DellEMC_XtremIO_iSCSI/3e8a769/DellEMC_XtremIO_iSCSI/641 : SUCCESS in 1h 35m 21s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"e644146de3b70b2c01dff25e242938fd7f06608f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-13 19:30:20.000000000","message":"Abandoned\n\nThis has served its purpose -- for now.","accounts_in_message":[],"_revision_number":4},{"id":"4d8f239d84c8b9dd8ff443740aa49a6c6b165733","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-13 22:22:38.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/15/745415/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a6befbfe3561ad03c60935befe31ec8c59c7fcb3","revisions":{"dca61b7cdfe3ee912e52dba644d4e482aaab907f":{"kind":"REWORK","_number":1,"created":"2020-08-07 20:49:21.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/15/745415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/745415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/745415/1"}}},"commit":{"parents":[{"commit":"074f6faf59b89eb5c27c87599b39a885815f4d8e","subject":"Merge \"Add lsscsi to bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/074f6faf59b89eb5c27c87599b39a885815f4d8e"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-07 20:47:03.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-07 20:47:03.000000000","tz":-240},"subject":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures","message":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures\n\nTesting a fix.\n\nDepends-on: https://review.opendev.org/745414\nChange-Id: I07a689c80aa08f98d1face108a9f379ddd7e8606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dca61b7cdfe3ee912e52dba644d4e482aaab907f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dca61b7cdfe3ee912e52dba644d4e482aaab907f"}]},"branch":"refs/heads/master"},"ba3e3bcbb9cf10df5168e4b6ac2bd9adf3dbdf09":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-07 21:01:39.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/15/745415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/745415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/745415/2"}}},"commit":{"parents":[{"commit":"ea745de6ca12876d12227b3d7723794bf9cb7980","subject":"Merge \"Fix revert to snapshot for non admins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea745de6ca12876d12227b3d7723794bf9cb7980"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-07 20:47:03.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-07 20:59:44.000000000","tz":-240},"subject":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures","message":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures\n\nTesting a fix.\n\nDepends-on: https://review.opendev.org/745414\nChange-Id: I07a689c80aa08f98d1face108a9f379ddd7e8606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba3e3bcbb9cf10df5168e4b6ac2bd9adf3dbdf09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba3e3bcbb9cf10df5168e4b6ac2bd9adf3dbdf09"}]},"branch":"refs/heads/master"},"f38a6b3e68dc34694a30f106afc532dc977757b3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-12 12:49:26.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/15/745415/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/745415/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/745415/3"}}},"commit":{"parents":[{"commit":"ea745de6ca12876d12227b3d7723794bf9cb7980","subject":"Merge \"Fix revert to snapshot for non admins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea745de6ca12876d12227b3d7723794bf9cb7980"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-07 20:47:03.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-12 12:49:18.000000000","tz":0},"subject":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures","message":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures\n\nTesting a fix.\n\nDepends-on: https://review.opendev.org/745414\nDepends-on: https://review.opendev.org/745752\nChange-Id: I07a689c80aa08f98d1face108a9f379ddd7e8606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f38a6b3e68dc34694a30f106afc532dc977757b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f38a6b3e68dc34694a30f106afc532dc977757b3"}]},"branch":"refs/heads/master"},"a6befbfe3561ad03c60935befe31ec8c59c7fcb3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-08-12 12:49:48.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/15/745415/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/745415/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/745415/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/745415/4"}}},"commit":{"parents":[{"commit":"ea745de6ca12876d12227b3d7723794bf9cb7980","subject":"Merge \"Fix revert to snapshot for non admins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea745de6ca12876d12227b3d7723794bf9cb7980"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-08-07 20:47:03.000000000","tz":-240},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-12 12:49:44.000000000","tz":0},"subject":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures","message":"DNM: cinder-tempest-plugin-lvm-lio-barbican failures\n\nTesting a fix.\n\nDepends-on: https://review.opendev.org/745838\nChange-Id: I07a689c80aa08f98d1face108a9f379ddd7e8606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6befbfe3561ad03c60935befe31ec8c59c7fcb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6befbfe3561ad03c60935befe31ec8c59c7fcb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
