)]}'
{"id":"openstack%2Fcinder~746619","triplet_id":"openstack%2Fcinder~stable%2Frocky~I05dfc996a635299990fdafba841ab6c4fd7d32c7","project":"openstack/cinder","branch":"stable/rocky","topic":"bug/1873738","hashtags":[],"change_id":"I05dfc996a635299990fdafba841ab6c4fd7d32c7","subject":"RBD: Cleanup temporary file during exception","status":"MERGED","created":"2020-08-18 00:12:11.000000000","updated":"2020-09-18 23:05:19.000000000","submitted":"2020-09-18 23:02:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746619-1600470158569-276e733d","meta_rev_id":"9686d1765ca01de3437d9bd455c4658f43d96b52","_number":746619,"virtual_id_number":746619,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-08-19 17:50:00.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2020-08-19 14:58:32.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-18 13:31:52.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-19 13:04:40.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 23:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-19 15:03:01.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-19 13:15:25.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-08-19 21:02:13.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-08-19 15:44:46.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-19 15:48:34.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"date":"2020-08-20 07:37:18.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-19 22:03:49.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-19 12:54:04.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 19:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-18 13:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-18 19:10:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 04:03:48.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-08-18 04:58:13.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-18 13:21:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-08-18 13:31:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-19 13:04:40.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-08-19 13:15:25.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-19 14:58:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-19 15:03:01.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-19 15:44:46.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-19 15:48:34.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-08-19 17:50:00.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-08-19 21:02:13.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-08-19 22:03:49.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-20 07:37:18.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 19:10:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-18 23:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02f6d39a4985e7e4f15442ccae4ca5c8c6086695","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-18 00:12:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"57a22d9ae26e5b1b2794e5f7ea9e04d747524558","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-18 00:17:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/19/746619/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e79a612 : FAILURE in 4m 30s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/19/746619/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/66603f2 : FAILURE in 4m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"1543c5ada8bda66402a83496a6d0b6db7bb5d2cb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-18 00:22:11.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/746619/1/iscsi/ : FAILURE in 7m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/746619/1/smb/ : FAILURE in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fbafc2bc8c6f6469c4e2390fc48ba4090462536f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-18 00:52:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/19/746619/1/check/dsvm-tempest-cisco-zonemanager-job/0e3da9e/ : FAILURE in 34m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2b4c6f8207bf551497c8b3ec028a43b925c44814","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-18 00:56:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/19/746619/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/29d31a3/29d31a3.html : FAILURE in 39m 47s","accounts_in_message":[],"_revision_number":1},{"id":"74bd1c809f0ff92e0c41a4df50ce55d114df354c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-18 01:14:53.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/19/746619/1/check/tempest-dsvm-full-bionic/d8c416d/ : FAILURE in 31m 36s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/19/746619/1/check/devstack-plugin-ceph-dsvm-bionic/6570369/ : FAILURE in 42m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cf147fcef34dc017f764a1045c29fd8e71f0eb19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 01:47:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f57714a27344f2cb669143473a7408d : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b5038d36c474a7ab0b85402fa56613d : SUCCESS in 10m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0795d21c70d5420ca931c0a84163ff4c : SUCCESS in 9m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/25b3171f22684200a8cb5a4be9fc7642 : SUCCESS in 8m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a27cd1918e9a476dbf4fde996878cc0f : SUCCESS in 1h 19m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/94e2905d80fa4236b1ef92266e21e144 : SUCCESS in 33m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73684268f6ab4d7fbe8570bb34794f85 : SUCCESS in 1h 14m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6422dec6c2c54535a28d6de987fe1fd7 : SUCCESS in 5m 40s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9714e1cec84647c88043a2309df3cf4f : SUCCESS in 1h 32m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f40004dbc8a5419094986df8bdd958ba : SUCCESS in 6m 11s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/14894b2790914ad282ce86e83d6fd362 : SUCCESS in 59m 26s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e7c15c64afb44677b9ba4b1e364eb4aa : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8737e7790db0406f94e0c66474b3c7a5 : SUCCESS in 7m 42s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3852a212655546d193b2d79af5b05e57 : SUCCESS in 5m 40s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/d4c7540e135144b9925fcab1efccd4bd : SUCCESS in 7m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e3da215c644a457990dd5d31f6359401 : FAILURE in 15m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d321ed12afcf446da270e8be872edca9 : FAILURE in 14m 38s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa432733ac814bba865a8fa0d58355ed : SUCCESS in 43m 56s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4493574624354eea9e8e215d8c7a1d0a : SUCCESS in 6m 16s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c9726098fe78454e8b2e3a8e5e6ed12e : SUCCESS in 1h 27m 05s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0df1d17f7b9b416eba296a039e6c8183 : SUCCESS in 51m 07s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/a72704e7fa0d413e865c556946e63be9 : SUCCESS in 1h 12m 29s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/23bd55c90c3a4f76bf2e1e87c66e02f3 : SUCCESS in 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d20d9b285d617b221201af3e4743f371fcfc796","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-08-18 01:51:44.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/19/746619/1/check/synology-dsm-cinder-iscsi/7879529 : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f7b6d344fd218b6d375846e04821f285f6e77c1d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-18 02:34:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/19/746619/1/upstream-check/cinder-cDOT-iSCSI/5617c94/ : FAILURE in 27m 42s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/19/746619/1/upstream-check/cinder-cDOT-iSCSI-IPv6/afc2adb/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/19/746619/1/upstream-check/cinder-cDOT-NFS/615ab1d/ : FAILURE in 26m 32s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/19/746619/1/upstream-check/cinder-cDOT-NFS-IPv6/85c012b/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/19/746619/1/upstream-check/cinder-cDOT-FCP/38d868d/ : FAILURE in 45m 44s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f1e5a66f9f7411026c71668c0a252af8604507e1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-18 02:38:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/19/746619/1/upstream-solidfire-check/cinder-solidfire-iSCSI/21ea2f4/ : FAILURE in 49m 54s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"1b0334a835e7b807eab0e534c0a8dc646715e53d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-18 02:47:37.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/19/746619/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2eb19299098b3ba65386a225d7daddb99adccfd9","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-18 03:58:20.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/19/746619/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a93d6708e6501bc868015e260e7488f039904b18","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-08-18 04:03:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"01f33836667ef975465daf3291dde27270878cd3","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-08-18 04:58:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/1a80d669f315de499c2bacc91fe13e9961c56ef4/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b8aa78ee0865191d35b686ebc5c1f831065f3b46","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-18 06:32:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/19/746619/1/check/dsvm-tempest-zadara-driver/4146716 : FAILURE in 1h 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25143fc1a4363125509fc75efe886ef943dc45b3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-18 09:50:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThe patch itself is a clean cherry pick from Stein, where the patch has merged. One small issue is with the commit message, though. Please fix that and the patch will be good to go.","accounts_in_message":[],"_revision_number":1},{"id":"9d875555fbf1af473812776320023ef8e7557baa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-18 09:51:31.000000000","message":"Topic changed from bug/1873738-stable/rocky to bug/1873738","accounts_in_message":[],"_revision_number":1},{"id":"c446494759819f827988da9277247a52476ae807","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-18 13:00:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"266c7b5730012d624f9899c870cb1b95a7d51617","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-18 13:01:29.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks Elod","accounts_in_message":[],"_revision_number":2},{"id":"2b79f513ae1909cc12091297823c9cbdab9c3dbc","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-18 13:05:44.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/19/746619/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/8f4613e : FAILURE in 4m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/19/746619/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6d22331 : FAILURE in 4m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"fa6673e909d3e406913832c2868da086bed34401","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-18 13:09:41.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/746619/2/iscsi/ : FAILURE in 8m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/746619/2/smb/ : FAILURE in 7m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e56ce80a2fb82665eeff6be80c7b5c8c97736e30","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-18 13:21:50.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for the update Rajat, now it looks good to me.","accounts_in_message":[],"_revision_number":2},{"id":"409c642a13eaf358739267827057aae57e29b7da","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-08-18 13:31:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/19/746619/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cb1a88c/cb1a88c.html : FAILURE in 26m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9ab8bce58873c6f794ea37beb996d9fc252bf292","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-18 13:35:56.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/19/746619/2/check/tempest-dsvm-full-bionic/b635bc4/ : FAILURE in 20m 46s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/19/746619/2/check/devstack-plugin-ceph-dsvm-bionic/b8e3bde/ : FAILURE in 28m 23s","accounts_in_message":[],"_revision_number":2},{"id":"dcd89aa5d0036fdc3393056b515274db1d0f29a5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-18 13:41:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/19/746619/2/check/dsvm-tempest-cisco-zonemanager-job/28ab2e8/ : FAILURE in 36m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4091ddbce4452858b15a196927f2bc944e2da44f","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-18 13:52:49.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/19/746619/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"466ff031e2d5b68330f38fc1e7b6d86698c300a8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-18 14:29:49.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/19/746619/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"cc654385092590d926f660864342c90217acf255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 14:54:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/675f76f9f58c424080e6ade850fe339c : SUCCESS in 9m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/29e2f1b83b5847eebdd8b893c4343616 : SUCCESS in 8m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7a06b06dee054bedb914a76328b6b72b : SUCCESS in 9m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f9b4a114e8594a84ba916b911f2128b0 : SUCCESS in 9m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5458709df0ff4878a54a1a41c25238c3 : SUCCESS in 1h 34m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/afde12d1e30b47e2ac0480161e5a9122 : SUCCESS in 1h 08m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd1e9ca267324e97a2ff21751168ad9c : SUCCESS in 1h 20m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e18ca14cc5a42b69eb00141e56803aa : SUCCESS in 8m 10s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bbea575d067d488f8bf0251288ad8846 : SUCCESS in 1h 40m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fee02554d42644d0aea641933aaa96b6 : SUCCESS in 11m 18s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/fc8b4a8b122747558f855e6dc2b482e3 : SUCCESS in 1h 00m 03s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fe2d3d287a3e46ada09985c7c2bc1ff3 : SUCCESS in 10m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9871bf20153430bbc202bc9b729f213 : SUCCESS in 12m 05s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/972a31bae54f4b7391731e623f47773b : SUCCESS in 7m 22s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/35fc219fb6c04bd3846c322ac4d0180a : SUCCESS in 10m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0983837c06ff4b05a310effcddae3a9e : FAILURE in 20m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd419b510ae24d38ac0da8e7b88b7b73 : SUCCESS in 25m 20s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1490e593a865469994ebcb7be4582356 : POST_FAILURE in 1h 12m 03s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b7a0209d76704fc189950522b73626ef : SUCCESS in 11m 19s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/371cc7bf68b74ec6a6e159af12ee714d : FAILURE in 9m 00s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1c66642cafef4067837396ffa8d185af : FAILURE in 7m 37s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/3421585f2fbc4e9ea261ac2d9e5d980a : SUCCESS in 1h 15m 17s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/fdaf165d79e84d65b90bf4063f7b9f5f : SUCCESS in 1h 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ad403b240648208e275895f4d9d54fbb9032185","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-18 14:57:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8c2e1c4417489d259982b55d5a9c3da0ee73f3b0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-18 15:02:14.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/19/746619/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c34d607 : FAILURE in 4m 32s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/19/746619/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/bfdf826 : FAILURE in 4m 32s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"2db47ea403e5916c46ac548fb9728e72cbcc4e7a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-18 15:08:10.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/746619/2/iscsi/ : FAILURE in 10m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/746619/2/smb/ : FAILURE in 9m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c92e2fa92723dd5194288a72dd05a4b87cda0f17","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-18 16:01:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/19/746619/2/check/dsvm-tempest-cisco-zonemanager-job/0b5ab74/ : FAILURE in 39m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1b7803df16628f5e3c293b0855f364e4f85a7a41","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-18 16:02:31.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/19/746619/2/check/tempest-dsvm-full-bionic/af9ac02/ : FAILURE in 24m 28s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/19/746619/2/check/devstack-plugin-ceph-dsvm-bionic/e468c23/ : FAILURE in 38m 10s","accounts_in_message":[],"_revision_number":2},{"id":"feedcd1f9e575edb24615e773bb536b6835dd70a","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-18 16:24:51.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/19/746619/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d158b05da910e9f08b7131a098529fd5d148965a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 16:52:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87b8ee91bea44d795d027e8e99db4ab : SUCCESS in 8m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd17b27392b942cda40d415714c6eaab : SUCCESS in 10m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/50f0df28b5cf4d3f980c0d21993fe7c3 : SUCCESS in 13m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/899a2c9d1dbf440088dfac65e298ef25 : SUCCESS in 14m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/80a322f8077241189ffe89068187b789 : POST_FAILURE in 54m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/74d3a7dcf5ed40bd811cbeaa66470ae5 : FAILURE in 6m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5718b32caaad4e57a1e14465bd1131ad : SUCCESS in 1h 17m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19838b6df83042c78cfb1596246971b1 : SUCCESS in 9m 45s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/012ce2d7de5341e1b623f8756b6c3d0e : FAILURE in 7m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9614614e9099411696e8a587484bdb9d : SUCCESS in 9m 37s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/f5eb20966dec4ae1aae53e3d14b5676e : SUCCESS in 1h 21m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e5185ca6e6144314be3ef4a8533ab585 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ebc95a3d499433ea7929987f59b0beb : SUCCESS in 10m 41s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2303b02a71364bdcb01221a33c898675 : SUCCESS in 9m 53s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/004f9a268c6346bdab4390e628392ae0 : SUCCESS in 11m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/76852fc805ae45d9b53389fa3cb80e8b : FAILURE in 22m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/465e9e0fb1f94ee59b1d6b841dedf05d : SUCCESS in 29m 41s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f0c345fd14ee466d831c17de2ab9e829 : SUCCESS in 1h 02m 47s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/118b736c47dd45d6a40552a09bda341c : SUCCESS in 7m 15s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8a3c009693584b55b7e6ae829b64dd74 : SUCCESS in 1h 16m 54s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0164862ae66348b99d3f01eefe004e19 : FAILURE in 6m 49s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/58e5b351573b4340a5972da8376fa917 : SUCCESS in 1h 18m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/d1567a7091b34142ad9f93b5ee54c2af : SUCCESS in 1h 24m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfbc0a20c70d58f075e073e09fbb00930971fd40","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-18 19:12:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/19/746619/2/check/dsvm-tempest-zadara-driver/da31bb5 : SUCCESS in 1h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"890f74502df2233da9fc1d871e2a4f37ee7dd985","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-19 12:54:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b353453e4c7ab84264b26a1b55019293ab0c92f3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-08-19 13:04:40.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/746619/2/iscsi/ : FAILURE in 8m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/746619/2/smb/ : FAILURE in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1d546dddecaabec9670b8a50299a42620fcd7ad6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-19 13:15:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/19/746619/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4d6d71c : FAILURE in 4m 27s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/19/746619/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/dff82ea : FAILURE in 5m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"3699456d2e718c4f73494f5ccede95c36597304f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 14:56:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/034ccf34fad54073b118ef689708b285 : SUCCESS in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/554ee5d32bb643d1a4717047d5f6fc4d : SUCCESS in 14m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f81e084377be4841bb497a10fc9398c8 : SUCCESS in 9m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bb5a5ceed64a4627a300a1dc30847d81 : SUCCESS in 9m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3e7678499a8d48079e15a58e34c2e1d2 : SUCCESS in 1h 36m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8449e5a5f31e4409abc7b89df5e2ae52 : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d9856ca697445219aebd21b5625880c : SUCCESS in 1h 31m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f6a43cfabdb4529a5b362212fb5c404 : SUCCESS in 8m 22s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/401a033984084d4182deb3815cc93d1b : SUCCESS in 1h 58m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/97fa8b365641437bbb346182f0b28482 : SUCCESS in 9m 14s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/e1a0cf246a814962bde48e82a0e35fa1 : SUCCESS in 1h 00m 35s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2fed55730f7d4e7baa474d8210242e34 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4479b524f03242438a8fe5c2e89c0324 : SUCCESS in 12m 03s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4e8c736f5bd46028e32dec7f83e6478 : SUCCESS in 6m 30s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/9cb34d96f1e14fae83272748d936a5be : SUCCESS in 10m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/21dea2ca8c2842d090772a988e490162 : FAILURE in 15m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c35a7233b0d4bc38ef48cfd5da6a0f0 : FAILURE in 23m 45s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7440a52e562d4657910fc583660d09fd : SUCCESS in 1h 14m 21s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/59aa50b2d9374693909202c7ac6c6f70 : SUCCESS in 8m 30s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4c7bbdaaaa214d06800dfe1f0fc7e57b : SUCCESS in 1h 11m 40s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cfa9ae761b6a471ab07d37c0547fb5ba : SUCCESS in 1h 38m 38s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/8112a4da55b1417b9cb9a697e7391ceb : SUCCESS in 1h 13m 50s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/0fadcf938c604249a85da49c51b2699c : SUCCESS in 1h 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"883db5dd0fad1ddaeb443e5f17beac84c1fe3f80","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-19 14:58:32.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/19/746619/2/check/tempest-dsvm-full-bionic/50b3dd6/ : FAILURE in 32m 53s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/19/746619/2/check/devstack-plugin-ceph-dsvm-bionic/3af8a8d/ : FAILURE in 33m 46s","accounts_in_message":[],"_revision_number":2},{"id":"90dc13ac2b1bc0737a15b9cc22bcef0018d20259","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-08-19 15:03:01.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/19/746619/2/check/synology-dsm-cinder-iscsi/2eafbe2 : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"d39f4c96e97da39a6c4d453e066e6083ba0dd6f0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-19 15:44:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/19/746619/2/upstream-check/cinder-cDOT-iSCSI/c9abb8a/ : FAILURE in 28m 49s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/19/746619/2/upstream-check/cinder-cDOT-iSCSI-IPv6/f3b4650/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/19/746619/2/upstream-check/cinder-cDOT-NFS/d384734/ : FAILURE in 29m 52s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/19/746619/2/upstream-check/cinder-cDOT-NFS-IPv6/bbd2904/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/19/746619/2/upstream-check/cinder-cDOT-FCP/11808f2/ : FAILURE in 46m 53s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4639188630985c7a7701c88a3766f74b6bd90986","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-19 15:48:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/19/746619/2/upstream-solidfire-check/cinder-solidfire-iSCSI/dd58290/ : FAILURE in 50m 25s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"e9754305246dfdf9b5aa31795569a8d632848d31","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-19 17:50:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/19/746619/2/check/dsvm-tempest-cisco-zonemanager-job/034435b/ : FAILURE in 39m 39s","accounts_in_message":[],"_revision_number":2},{"id":"8041a7ab06da827dad7e4c54cd7b01952b87287b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-08-19 21:02:13.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/19/746619/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"adf1db0c902f4902c8d0debffac744821a998de7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-08-19 22:03:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/746619/2/20200820050625/NEC-ISCSI : SUCCESS in 1h 56m 46s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/746619/2/20200820050625/NEC-FC : SUCCESS in 1h 57m 21s","accounts_in_message":[],"_revision_number":2},{"id":"1250e19cac3725beaf84a32ed642559c33b01ecc","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-08-20 07:37:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/19/746619/2/check/dsvm-tempest-zadara-driver/5810085 : FAILURE in 1h 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd33dced3095523a66b5a5f357d36c4d462d6bb5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-18 19:10:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMerged to more recent branches, good to backport.","accounts_in_message":[],"_revision_number":2},{"id":"dd259c373246cfe985ea170c66f012a96f3cf00e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 19:11:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e39e9b3a048d48cfbb71a1c33e55dabd41d6ede4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 23:02:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"815c395bea842a3f70d4b03bedc67241163efaeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 23:02:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f32a57d34024e6aa00adb5d071d35be : SUCCESS in 8m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fda623679eff4d43abf7b26ab3033b3d : SUCCESS in 9m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2acf15abbc2d4cc6badb854a9955fd56 : SUCCESS in 9m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ee506877b8c2456f9124afc8a9cf40af : SUCCESS in 8m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c05a3a756bc14e0b9973efdae585a525 : SUCCESS in 1h 41m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cfa1f016674c4d1c8773679785e00bb4 : SUCCESS in 43m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d0da794d5494b39b74fad9b14add1ef : SUCCESS in 1h 14m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c132f34c51b94ba193a9678a086eabcb : SUCCESS in 8m 03s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/68d80587519e4df789f7c10dc17ab1e4 : SUCCESS in 1h 02m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/03105fd1ae9f44dd854098d5c0da193d : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9638c7d0e95d44728b76870178f5f438 : SUCCESS in 9m 35s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6623ddbb4d884ae9bea5769fc766f414 : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":2},{"id":"9686d1765ca01de3437d9bd455c4658f43d96b52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 23:05:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8874e84130c6473b97457a0d9aff8861 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b38df60b191230cd9e75eca9603f3a1c15f646e","revisions":{"1a80d669f315de499c2bacc91fe13e9961c56ef4":{"kind":"REWORK","_number":1,"created":"2020-08-18 00:12:11.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/19/746619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/746619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/746619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/746619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/746619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/746619/1"}}},"commit":{"parents":[{"commit":"44b64b1ba2b8289e66d43fe79883030d830d4bd5","subject":"Merge \"VNX: delete the LUN from VNX backend\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44b64b1ba2b8289e66d43fe79883030d830d4bd5"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-20 09:00:39.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-18 00:12:11.000000000","tz":0},"subject":"RBD: Cleanup temporary file during exception","message":"RBD: Cleanup temporary file during exception\n\nDuring copy image to encrypted volume operation, we create a\ntemporary image file to convert source image into an encrypted\nimage and rename it to source image name.\nIf any error occurs during this process, there is no code to\ncleanup the temporary \u0027.luks\u0027 file.\nThis patch adds the following cleanup.\n\nConflicts:\n    cinder/image/image_utils.py\n    cinder/tests/unit/volume/drivers/test_rbd.py\n\nChange-Id: I05dfc996a635299990fdafba841ab6c4fd7d32c7\nCloses-Bug: 1873738\n(cherry picked from commit 7c95f3969f4c0df5f818f96cb662dddc17a8dd6f)\n(cherry picked from commit 02de7897d4e4ec3be6c4684a2e73a054956d1a00)\n(cherry picked from commit 48774c89abe6f9770a55a4f8be9f682b22b900d1)\n(cherry picked from commit 175eb628135ecd828a9636952f0dcf95b6a76fed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a80d669f315de499c2bacc91fe13e9961c56ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a80d669f315de499c2bacc91fe13e9961c56ef4"}]},"branch":"refs/heads/stable/rocky"},"5b38df60b191230cd9e75eca9603f3a1c15f646e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-18 13:00:57.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/19/746619/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/746619/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/746619/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/746619/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/746619/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/746619/2"}}},"commit":{"parents":[{"commit":"44b64b1ba2b8289e66d43fe79883030d830d4bd5","subject":"Merge \"VNX: delete the LUN from VNX backend\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44b64b1ba2b8289e66d43fe79883030d830d4bd5"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-20 09:00:39.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-18 12:58:50.000000000","tz":0},"subject":"RBD: Cleanup temporary file during exception","message":"RBD: Cleanup temporary file during exception\n\nDuring copy image to encrypted volume operation, we create a\ntemporary image file to convert source image into an encrypted\nimage and rename it to source image name.\nIf any error occurs during this process, there is no code to\ncleanup the temporary \u0027.luks\u0027 file.\nThis patch adds the following cleanup.\n\nChange-Id: I05dfc996a635299990fdafba841ab6c4fd7d32c7\nCloses-Bug: 1873738\n(cherry picked from commit 7c95f3969f4c0df5f818f96cb662dddc17a8dd6f)\n(cherry picked from commit 02de7897d4e4ec3be6c4684a2e73a054956d1a00)\n(cherry picked from commit 48774c89abe6f9770a55a4f8be9f682b22b900d1)\n(cherry picked from commit 175eb628135ecd828a9636952f0dcf95b6a76fed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b38df60b191230cd9e75eca9603f3a1c15f646e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b38df60b191230cd9e75eca9603f3a1c15f646e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
