)]}'
{"id":"openstack%2Fcinder~750782","triplet_id":"openstack%2Fcinder~master~I06398428f82b3617a0e8b04d86cd6a839b143263","project":"openstack/cinder","branch":"master","topic":"bug/1895035","attention_set":{},"removed_from_attention_set":{"28801":{"account":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"last_update":"2022-12-01 13:29:10.000000000","reason":"removed on reply"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"},"1736":{"account":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"},"26537":{"account":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"last_update":"2023-02-21 12:09:47.000000000","reason":"Change was marked work in progress"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-02-21 12:09:47.000000000","reason":"Change was marked work in progress"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"},"32238":{"account":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2022-12-01 13:54:43.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2022-09-29 09:30:38.000000000","reason":"\u003cGERRIT_ACCOUNT_30615\u003e replied on the change","reason_account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}},"24921":{"account":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"last_update":"2023-02-21 12:09:47.000000000","reason":"Change was marked work in progress"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"},"26077":{"account":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"last_update":"2022-11-25 13:40:16.000000000","reason":"removed on reply"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"},"32159":{"account":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"last_update":"2023-02-21 12:09:47.000000000","reason":"Change was marked work in progress"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-10-27 18:44:36.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I06398428f82b3617a0e8b04d86cd6a839b143263","subject":"WIP RBD: Fix Restore Full Backups","status":"ABANDONED","created":"2020-09-09 18:31:02.000000000","updated":"2023-04-24 16:43:16.000000000","total_comment_count":51,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"1c4fe0d6ed4e34ec95e272cc5da8c2da049f9760","_number":750782,"virtual_id_number":750782,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-20 15:12:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2020-09-09 20:36:57.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-10 02:29:22.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-10 05:54:57.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-05 21:52:02.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-05 23:02:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 02:19:54.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 02:48:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 05:10:09.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 22:37:03.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 06:17:57.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 21:46:45.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-10-22 11:25:15.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2020-11-20 02:56:01.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 02:57:38.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 02:58:39.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 03:24:04.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 03:56:12.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-11-20 04:08:56.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-11-20 04:35:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 04:54:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 05:57:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 05:59:41.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 06:00:20.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"REVIEWER"},{"updated":"2020-11-20 08:11:36.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 08:19:00.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-11-20 08:46:45.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-11-21 17:32:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-08 15:44:25.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2021-01-20 09:40:51.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-20 10:19:21.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-01-20 10:57:50.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-01-20 15:11:08.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-07 14:21:16.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-04-12 14:31:57.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-08-10 06:28:33.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-08-18 19:23:19.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-09-02 21:02:30.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-09-03 13:43:53.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-05 06:11:01.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-09-05 22:35:32.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-11-13 02:55:00.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-11-19 15:05:58.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2022-01-10 20:22:32.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-03-10 22:29:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-03-11 16:25:30.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-03-11 22:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2022-03-14 09:41:20.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-03-23 14:54:31.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2022-03-27 09:43:16.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-03-28 01:24:48.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-03-30 00:41:24.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-04-25 13:23:07.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-04-25 21:51:51.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-05-23 20:29:32.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-11-15 22:05:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-11-29 22:58:45.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-01-23 14:16:16.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-04-20 16:20:41.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"}],"messages":[{"id":"9a94afbdd9a8437dab1ae89a6ba6b18b0373efa0","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-09-09 18:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a550ef3e59fca4b9fda23158849049bf624f34a6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-09 18:36:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/82/750782/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/98149fe : FAILURE in 4m 52s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/82/750782/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/2ce6bcc : FAILURE in 5m 29s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"e9aa581ddba0e22b66f0b074511704fd643468f3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-09 18:46:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/82/750782/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/37da650/37da650.html : POST_FAILURE in 7m 23s","accounts_in_message":[],"_revision_number":1},{"id":"794445c8a64cb29aa23ebaf887c4a99d01911d25","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-09 18:55:31.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/d90deb03022d49da0bb02bdf993a4a042a0a3155/index.html : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"43f4de801d9c31ba64342614154de596925a68de","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-09-09 19:17:07.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/1/iscsi/ : FAILURE in 25m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/1/smb/ : FAILURE in 45m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5028702853cb2a2506616ddb31462d43955494bc","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-09 20:36:57.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/750782/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f89016b9bf267660d1b26debfad3807c6bfcdd0c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-09 21:03:29.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/82/750782/1/check-not-vote/ext-cinder-zuul/aa30173 : SUCCESS in 36m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5d0f13dfdc643786e80738e1b2eaa208cb7473a4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-09 22:08:45.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-1/ns5_nfs : SUCCESS in 2h 30m 51s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-1/ns5_iscsi : SUCCESS in 2h 26m 30s","accounts_in_message":[],"_revision_number":1},{"id":"690a469d58e58d41f2780cef221c52eb6cbe60d5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-09 22:54:08.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bf8fce53cc3c0273b6cd4a5e1b11246c9c2a1d39","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-09 22:56:18.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/1/check/tempest-dsvm-full-bionic/591e818/ : FAILURE in 1h 45m 54s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/1/check/devstack-plugin-ceph-dsvm-bionic/8555e3d/ : SUCCESS in 2h 03m 38s","accounts_in_message":[],"_revision_number":1},{"id":"87f8648a7dc41e3b319de681e0face05c2eb35aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:49:51.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8e50a6aee0a4655b3b56c9f3e2cd86c : FAILURE in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00123126c8fe4adeb02c63a80e6204aa : FAILURE in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25b52f5d12314231852d41bc130c4b87 : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97b7fdbd333a4de4a4cb1afdd0f960a8 : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b32e45c3f37488e9abb2a26d1b1b83a : TIMED_OUT in 31m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/67a4ca8673d34a36ae30b8fde8569a01 : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ac622362538a40f4ad88d42471bc4708 : FAILURE in 1h 40m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/761510ec47cc4f34a4ad6bbdfba77ec3 : SUCCESS in 16m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cab82d5d6a13411e8a4e24d2392aba6f : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ad7fa6e50864a9587d6b579687d3a97 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fbb33414eff4472884756374d4646ecc : SUCCESS in 8m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5f9040f5333340c7830bf8fd77fca5e6 : SUCCESS in 42m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4d55b5dab25a405dbd4e1bd411a5eb11 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3354e4be0295463da87092e4545e75cf : SUCCESS in 1h 10m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9e7be1317b794cc5b69900799dc4cfe3 : FAILURE in 1h 12m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/570826c648d7436395702483a0429fa8 : SUCCESS in 1h 18m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c3053371a2f4ec6be81208ea441f723 : SUCCESS in 57m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5bf593a78204fa9bd8a0cd4b32e5d3a : SUCCESS in 1h 23m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/38fbaa454c55440e80a3deb588d2866f : SUCCESS in 1h 20m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e1a4388d50ed45e5a5ec2f3ee3287340 : SUCCESS in 1h 59m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aa0ab34b3d04686b73e7d86cb53f18b : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5b1df7edfb809799dc74d1f4e1a23ce5ea881e01","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-09 23:57:50.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b7cc864f9602f20d3e4f32daa128ef8c67e5d071","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-09-10 02:29:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/82/750782/1/check/dsvm-tempest-zadara-driver/18e3e17 : FAILURE in 1h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5a64abf9618cefdcd2f2d114e0ed3aecb54710c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-09-10 02:36:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/82/750782/1/check/cinder-storpool-tempest/5a8ed87/ : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0f75118456cdfa412a25d978948a953dc1795ba6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-09-10 05:54:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/82/750782/1/check/DellEMC_XtremIO_FC/c392d8a/DellEMC_XtremIO_FC/847 : SUCCESS in 1h 34m 08s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/82/750782/1/check/DellEMC_XtremIO_iSCSI/8504cd8/DellEMC_XtremIO_iSCSI/811 : SUCCESS in 1h 38m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"6ac8eae74e8ef51ced6afaf5b812b826f4ac0a23","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-10 09:50:58.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"dd4117841f15502836f1f7cae8f63c39c3b04022","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-10 10:44:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/1/20200910171958/NEC-FC : SUCCESS in 2h 19m 15s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/1/20200910171958/NEC-ISCSI : SUCCESS in 2h 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"735be56c2432b908228a1fbbcea92d0374fe4f1f","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-10-05 21:51:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4f52c88e4cc63aeeece811b259cc9d3e042bf831","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-10-05 21:52:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/82/750782/2/check/fujitsu-eternus-dx-iscsi/03331e0 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/82/750782/2/check/fujitsu-eternus-dx-fc/b634e03 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"040f435a5e1a3448921868053a88d724650ab8b0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-10-05 21:52:20.000000000","message":"Patch Set 2:\n\nStarting cinder-check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e084e8ed0be28ceb8a3a65a81529e3359ab9558","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-05 22:15:08.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/b58e0a16a51cbbf925bdc7162f7ec659e6e4f563/index.html : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"08b8585e5c3b84c28f89f61fd4bed77f7599e168","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-05 22:33:45.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/82/750782/2/check-not-vote/ext-cinder-zuul/4359e77 : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5f003a441d882ee1a29972235e99e215647cacba","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-10-05 23:01:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/2/ : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":2},{"id":"aa75d7ab95acb8e321746017bd2cfab3cc733358","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-10-05 23:02:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/82/750782/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/34b404f/34b404f.html : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a59de05d65139cf31393f86c8b97986b7c7ee378","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-05 23:31:53.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/2/check/tempest-dsvm-full-bionic/0d8c2b3/ : FAILURE in 1h 36m 50s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/2/check/devstack-plugin-ceph-dsvm-bionic/cfe01e3/ : FAILURE in 22m 21s","accounts_in_message":[],"_revision_number":2},{"id":"6f7e611347323e50a4401b5b88eb0264fa348d98","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-05 23:42:11.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e857e2f14eb4a76c9c2ef0a9aee77ffe1b388c8c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-05 23:52:51.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0d7b19dbcf61a0278f37462ef2f0c840d6968eda","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-06 01:57:58.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-2/ns5_nfs : FAILURE in 1h 30m 49s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-2/ns5_iscsi : FAILURE in 1h 28m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8505bc16e9a971ec4c36051a617a70644d699727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 02:00:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/487dd59ffa734de0b34268347aebc6dd : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3102a2b1aff4bc48a79337a5b98784e : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6470a1a921f34b88962fbc4b8677ae73 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fbf5859383c412eaec0fb2c168a60e4 : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2479f7b2cee41438b1dac05cd7ead9e : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/b8217311838f4b4581f3007cb9d0d41b : SUCCESS in 1h 05m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/093cae899e7f4a0c8cfaa8e933f3439f : SUCCESS in 1h 07m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1240ab6fec374c1e8cfc63ba5597db55 : SUCCESS in 11m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3533536fdab643c495b599ad1c07165a : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8ce9d38badd47938642a3b67835aee0 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/70ad993bc4cb47a39cbcd27f0d070b34 : SUCCESS in 4m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f8a8be78b21481f827b0cee63908a44 : SUCCESS in 48m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/00772dd6f48a469792989a979a7bcfbc : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3fffe49502ed4c5fa6c980d98220df96 : SUCCESS in 56m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e10ce08d1c9a49aea73bbe40dca5f8fa : FAILURE in 1h 20m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1d90d3f9ac644ed97901d6d18b5c4d2 : SUCCESS in 1h 23m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/456c6e14dfb54012a649f1c466e328c4 : SUCCESS in 1h 06m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/de32d4d064eb4f72afe85aac2f02a32f : SUCCESS in 1h 10m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9a0a3de7e2684e59946636c6f7d8af12 : SUCCESS in 1h 26m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0815588e1e2c4cdfa66d5720856e435a : SUCCESS in 2h 07m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb84ee6261ab460aa0e15a0ea9be6d17 : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":2},{"id":"82aa16383e7efc1cd2824f3fa26ba3aa3d5d9655","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-10-06 02:08:35.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/82/750782/2/check/synology-dsm-cinder-iscsi/03489d2 : FAILURE in 7m 46s","accounts_in_message":[],"_revision_number":2},{"id":"29a616fad1004217ce8413419c485df4c686a649","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-10-06 02:19:54.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-750782-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"10893fc9af3a1edc334c988c8ab7a4a78c662b43","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-10-06 02:48:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/82/750782/2/check/cinder-storpool-tempest/a4a78c7/ : SUCCESS in 1h 30m 51s","accounts_in_message":[],"_revision_number":2},{"id":"37827481b361d69805674f5b5bffca396062863c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-10-06 03:08:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/750782/2/upstream-solidfire-check/cinder-solidfire-iSCSI/1f41861/ : SUCCESS in 1h 06m 49s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"945b83194523efac3c61c901e987104939cf4bed","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-06 03:23:05.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"9aab4ad1955450445669d2685bf0fe714ca41c1c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-10-06 05:10:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/2/20201006112753/NEC-FC : SUCCESS in 2h 25m 54s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/2/20201006112753/NEC-ISCSI : SUCCESS in 2h 42m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a11515c3ccbdbf5ff04dce078aaef7c6ae01a1e5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-06 05:37:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/750782/2/upstream-check/cinder-cDOT-iSCSI/0d4ffe9/ : SUCCESS in 36m 12s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/82/750782/2/upstream-check/cinder-cDOT-iSCSI-IPv6/4c78700/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/750782/2/upstream-check/cinder-cDOT-NFS/a9c93e5/ : SUCCESS in 35m 48s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/82/750782/2/upstream-check/cinder-cDOT-NFS-IPv6/1837ec5/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/750782/2/upstream-check/cinder-cDOT-FCP/51574d6/ : FAILURE in 3h 36m 28s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"695094664dd953faca94f2a0107c2fbbee1b47e6","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-10-06 22:37:03.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/82/750782/2/check/nimble-iscsi-driver-dsvm-volume/2ca0a8f : FAILURE in 1h 54m 09s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/82/750782/2/check/nimble-fc-driver-dsvm-volume/05e290c : SUCCESS in 3h 09m 18s","accounts_in_message":[],"_revision_number":2},{"id":"75ecb8ac7424cccd8cb4f1d16554c01666f1b45a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-07 06:17:57.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-2/ns5_nfs : SUCCESS in 9h 5m 4s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-2/ns5_iscsi : SUCCESS in 8h 53m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a520cfd3731d329007eef6627190618e68bbe269","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-10-22 11:25:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nThank you for fixing this bug!!  We\u0027ll definitely have to backport it when it\u0027s merged.\n\n-1: Missing release note","accounts_in_message":[],"_revision_number":2},{"id":"db9f643748142d99f570715b2435ef661f161c23","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-27 23:16:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/82/750782/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/667562d : FAILURE in 11s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/82/750782/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/cf30565 : FAILURE in 20m 49s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"84a3bb85f717c689f251e49c633f71ef3bded5c4","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-11-19 19:57:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2182050fcf2dc151f09bca7dd22d94111e3fb4dc","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-11-19 20:15:07.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/ab6cd2a0a47467275eaa1eb24d7a3fe15c5c949e/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"dcbf15224944bd62c3fb9ce0c8fc0b376720050a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-19 20:17:33.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/82/750782/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/95e899c : FAILURE in 20m 01s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/82/750782/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/db81c2d : FAILURE in 19m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"08ce1df68f36db6e54a09da42a5749c26a05b439","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-19 20:37:49.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/82/750782/3/check-not-vote/ext-cinder-zuul/f6afdde : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7a82eac9ca3077c102584ea535f715abf21bc689","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-19 21:03:35.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/3/iscsi/ : FAILURE in 1h 05m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/3/smb/ : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":3},{"id":"447a3350efe0a09c2db9e6b07b46a8505f0add4b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-19 21:16:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/750782/3/check/dsvm-tempest-cisco-zonemanager-job/0fb329c/ : SUCCESS in 1h 14m 21s","accounts_in_message":[],"_revision_number":3},{"id":"77651338f9101ff5deff7f0de5467786e321a3b2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-19 21:56:45.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/3/check/tempest-dsvm-full-focal-py3/63d0dee/ : SUCCESS in 1h 32m 51s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/3/check/devstack-plugin-ceph-dsvm-focal-py3/ae48fbf/ : SUCCESS in 1h 52m 15s","accounts_in_message":[],"_revision_number":3},{"id":"569d537ca376fc901ed261f6d3df626547f063af","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-11-19 22:29:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/a9620b29a38748c0a81bfd54672b673b : FAILURE in 15m 31s","accounts_in_message":[],"_revision_number":3},{"id":"e70cfe1df8d72c4e2a9939c9c2c4231d330033fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-19 23:54:07.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/750782/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"f7a3d42447881a66ba8e6d7b24a3cc81a4bc5788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:39:31.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6aa128888983451e87bdfe6518c2bdf0 : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d11136e75714dd19fc22f20cb1faa0e : FAILURE in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9833b0f7f8f24a938f3de94ad3b8b320 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0371ffd8cfbb4bbfb89c8da9dcb3a85a : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0e1222be91d43c38651ae7a5a53b899 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d06ae8593564bd3b355f20626772fbb : SUCCESS in 14m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/450c67dfa46b447ca7dd4060618f76ac : SUCCESS in 1h 13m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36266cca1021499787640509e4e07b4b : SUCCESS in 1h 28m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3a3593279214ac2ac16a99aac7b29df : SUCCESS in 7m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/32d2372d287b4f57b70bdaba977f9055 : SUCCESS in 17m 17s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/658ebc5d21b94aa88b31b0db7d76c0ce : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97e2f32e7fc74c7a8a54bf53e293d7eb : SUCCESS in 6m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f03fac2f0a841cf8aaefcdc71ad58a5 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5f83f8a928f4b4bb2ea6cd07880fe1e : SUCCESS in 42m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b97872dae504d18a8255b50777bdb79 : FAILURE in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/187710beec214d9fa94b354d3ee43c2a : SUCCESS in 1h 28m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4d53292d7d41401d90c3d199fcaf8957 : FAILURE in 1h 23m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c2fb18606e9d4fe8ac728fe4e200090f : SUCCESS in 1h 31m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7be8bd7a15ab491b8271b781558d2d62 : SUCCESS in 1h 16m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/23f4b30aaa6e43aa93efcec0b5f10285 : SUCCESS in 1h 23m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0e09a5a150584893b3e40db37e8066d0 : SUCCESS in 1h 45m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46f9382c2a2d423fb3ed440216a83281 : SUCCESS in 2h 10m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f6dc8e4bbc34a658619ae64693e4a1f : SUCCESS in 1h 15m 59s","accounts_in_message":[],"_revision_number":3},{"id":"bfdfb9dbafe48d4ccbe125146fc20493da17418c","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-11-20 02:37:56.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6abdaeab6d8ed0f74f8addfc4290d582598573c6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-11-20 02:56:01.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/b369ae2f63b9463b8c69de594365e5a9 : FAILURE in 17m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d9b0be8344d9310414c56ac38c3ca7a23b8a9c5a","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-11-20 02:57:38.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/25ea0c72afec77a89c9149329f62d829c7f60c02/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"811ec54bbf2ed403c826499c59cafac276c5ab72","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-20 02:58:39.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/82/750782/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/cdd6e16 : FAILURE in 20m 20s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/82/750782/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/915c6f8 : FAILURE in 19m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"6e8a2a1b7aee4ff303bd9d02bde3216b6bec512c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-20 03:24:04.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/82/750782/4/check-not-vote/ext-cinder-zuul/778502f : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":4},{"id":"51dafbdf8a3811d236ce2ea5a8e486eb10875758","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-20 03:56:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/750782/4/check/dsvm-tempest-cisco-zonemanager-job/9560b83/ : FAILURE in 1h 13m 40s","accounts_in_message":[],"_revision_number":4},{"id":"fd223093c6e20b9daa26e2c5cbac5fc87513b434","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-20 04:08:56.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/4/iscsi/ : FAILURE in 1h 30m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/4/smb/ : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":4},{"id":"875099ef0ccbf859c9e47cfadb17648eb1873be1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-20 04:32:33.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"149e8b788a56875fc48e944a5a7f1037b3039a26","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-20 04:35:09.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/4/check/tempest-dsvm-full-focal-py3/c8e269f/ : SUCCESS in 1h 38m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/4/check/devstack-plugin-ceph-dsvm-focal-py3/33dd7ec/ : SUCCESS in 1h 47m 33s","accounts_in_message":[],"_revision_number":4},{"id":"33ef8992547ad42364bc8e5e9c84f39330a3c33e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-20 04:52:36.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/750782/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"bf3f0f850171f27cef682af417cdd5e3947c255e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 04:54:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b8de9a4baea44b982c92c88e4da30aa : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19fc35d64d714f40a6c58df60a1d69aa : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d450e69753c7472c95785c37256b93f9 : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33e5df7a2ddb461a9b2295525941e66b : SUCCESS in 12m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8ba93a8a0d44c62b351c12fd37e970d : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af6477366e34e02bc1c20d2af4177e9 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/f3a648c04a394929bd30114c05e60ec9 : SUCCESS in 1h 07m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92f611be554e4a41ae09e13cce9a829a : SUCCESS in 1h 14m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad388313de7443609ce90337dce15109 : SUCCESS in 6m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/99d0ec0913764316a9ad4accccd3b601 : SUCCESS in 13m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/148f45d43edc43d3a90b1771d419c7f6 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/caad36934570447cbcf85bdf8906f90d : SUCCESS in 5m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/95a2461ef0474caba1b0bbb44cacedbf : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e720108b1d8c43fc8aa910d15eeaa0cf : SUCCESS in 39m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0fcbee8c453843618cc3562ecc3d1e67 : FAILURE in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a2b0a4d80d2646f1b69c8de802cdce4a : SUCCESS in 1h 19m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cde7d0c8681d4285988aa6b6493f979a : FAILURE in 1h 27m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d3e932e43e2f4c079ba148646c7ec39b : SUCCESS in 1h 31m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bbee996f3242457dad6ee9ced43b497d : SUCCESS in 1h 37m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5dd5fab76a814816ba7007eb35da83db : SUCCESS in 1h 23m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ef7422bfa856487eb5f0a24669c9f325 : SUCCESS in 1h 21m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/359d581808ea427193b8ec3aae5e0f8d : SUCCESS in 2h 10m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c58c9ba192a44ca2b8fd735bf13222a5 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":4},{"id":"77506974b6f94ed19857155f01f3a7a1ec3bbad8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-20 05:48:03.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"032125f561d11324142aaa8fc5e34219d4dc14f2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-20 05:57:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/750782/4/upstream-check/cinder-cDOT-iSCSI/111fc11/ : POST_FAILURE in 38m 06s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/82/750782/4/upstream-check/cinder-cDOT-iSCSI-IPv6/22e37a9/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/750782/4/upstream-check/cinder-cDOT-NFS/ef354e9/ : POST_FAILURE in 42m 10s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/82/750782/4/upstream-check/cinder-cDOT-NFS-IPv6/4170792/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/750782/4/upstream-check/cinder-cDOT-FCP/0a7a09e/ : POST_FAILURE in 1h 02m 03s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"85e427f68aead4e68ee19e5e0c0749faadbaff30","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-11-20 05:59:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/82/750782/4/check/synology-dsm-cinder-iscsi/5fd8f87 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":4},{"id":"f448dc553cb44f0e561afaf612875c4c47068b64","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-11-20 06:00:20.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/82/750782/4/check/DellEMC_PowerStore_iSCSI/251cf60/DellEMC_PowerStore_iSCSI/127 : SUCCESS in 1h 05m 52s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/82/750782/4/check/DellEMC_PowerStore_FC/e29be6a/DellEMC_PowerStore_FC/123 : SUCCESS in 1h 05m 25s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"4b1d76b66e1a02acacc3b77311a67658aa776490","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-20 07:23:21.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"1c4fb8419069f13e23d59ca0cc15059ea9e0085c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-20 08:11:36.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/750782/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"36455d195b42d4b19981e240167bbdc0819ca89d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-20 08:19:00.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/82/750782/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"8fde430f6a391347b903912b036f4903078c5d0a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-11-20 08:46:45.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/750782/4/upstream-solidfire-check/cinder-solidfire-iSCSI/360a3b9/ : POST_FAILURE in 1h 50m 47s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"8fd5be7e5ab0c7bab3c0329dc3d5a65678513f74","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-21 17:32:05.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/82/750782/4/check/DellEMC_Unity_FC/2aaec5e/DellEMC_Unity_FC/174 : FAILURE in 3s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/82/750782/4/check/DellEMC_Unity_iSCSI/9bb591e/DellEMC_Unity_iSCSI/154 : SUCCESS in 1h 47m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"d0304e94441c4c105eaade148ff932515a5e30d4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-21 18:48:20.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/750782/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c75f397f684747fb8f2d3333faa6d44d2eb12968","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-22 09:26:40.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-3/ns5_nfs : FAILURE in 56h 47m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-3/ns5_iscsi : FAILURE in 56h 48m 24s","accounts_in_message":[],"_revision_number":3},{"id":"79a9bebd3aa8787534c40eced54b8db5f5e71cb7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-22 15:10:44.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-4/ns5_nfs : FAILURE in 1m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-4/ns5_iscsi : FAILURE in 2m 1s","accounts_in_message":[],"_revision_number":4},{"id":"f2a2e03a45344213930df8a325fb2d35541d06bd","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-11-23 00:02:21.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/82/750782/4/check/DellEMC_XtremIO_FC/46d1a82/DellEMC_XtremIO_FC/155 : SUCCESS in 1h 41m 01s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/82/750782/4/check/DellEMC_XtremIO_iSCSI/7ca3ca2/DellEMC_XtremIO_iSCSI/152 : SUCCESS in 1h 33m 39s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"e55910cba40d0f7ae70db49920b7bf4813d6ea30","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-23 03:33:27.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/750782/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"ba31a93afc0715ad5ced4ca100470e7d0262c2af","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-24 17:05:10.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-4/ns5_nfs : SUCCESS in 3h 43m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-4/ns5_iscsi : SUCCESS in 3h 40m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"95b21656f31ba4858008b353fee6d29d16b0fe98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-24 19:01:31.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/750782/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"4930803ced21ac2bbd561b6f0283911ea61c3628","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-24 20:48:35.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-4/ns5_nfs : SUCCESS in 3h 53m 27s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-4/ns5_iscsi : SUCCESS in 3h 41m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"70f1631096d6bbb05f54f03d24a5610ff4048ef7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-25 00:51:32.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/750782/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"58d54f6d5e3d4dd5eba7b451c19ee1230271b47c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-11-29 02:41:48.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/4/20201129074540/NEC-ISCSI : SUCCESS in 2h 13m 00s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/4/20201129074540/NEC-FC : FAILURE in 3h 56m 06s","accounts_in_message":[],"_revision_number":4},{"id":"dfaccf7aed9665b97a87e5c3c0c033bbe189921b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-01-08 21:37:18.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"eae2c39ee1f3fc7b623c626da2be354e45a61ac5","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-01-14 10:36:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nThanks for the fix!\n-1: Please add a unit test to prevent regressions","accounts_in_message":[],"_revision_number":4},{"id":"077362ee5bbda7d50f32c8732218629a0ecc0700","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-01-18 22:04:22.000000000","message":"Patch Set 4: Code-Review-1 Workflow-1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"86445dbba8d91371a58253d1de0a33453e427d5b","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-01-18 22:05:04.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review-1 Workflow-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e I\u0027ll add tests","accounts_in_message":[],"_revision_number":4},{"id":"c5a15d0146e43492544c281e75c671651af363d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-01-19 21:32:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae70dfea41b8e257587effc93b037b0cb5939301","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-19 21:32:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0aa7e1eb92dd2706505ddf767bc64ad1fe9e4ca2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-19 22:16:14.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/750782/750782/5/check-not-vote/ext-cinder-zuul/8cfa49e : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":5},{"id":"049d2b0c6a86e31500e7b9a1c818688dabe2c009","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-19 22:39:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/5/iscsi/ : SUCCESS in 1h 00m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/5/smb/ : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":5},{"id":"8d10e4a32365c47a85763873782e08ce7d1aefa9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 22:41:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/82/750782/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/458e12d/458e12d.html : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":5},{"id":"31b54680bc94fe4aef1e37479e131cd1e7fe62ec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 23:08:39.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/82/750782/5/check/iscsi-driver-primera/8f0516f/8f0516f.html : FAILURE in 1h 34m 43s","accounts_in_message":[],"_revision_number":5},{"id":"2dd3e84a71bef56bada3b1b8df5426c70b84163b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-19 23:29:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/750782/5/check/dsvm-tempest-cisco-zonemanager-job/00c9fcb/ : SUCCESS in 1h 20m 56s","accounts_in_message":[],"_revision_number":5},{"id":"c1a7bf94baa26dde7bd9d868e619ab792ca30fe8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-20 00:45:31.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/5/check/tempest-dsvm-full-focal-py3/01b3d99/ : SUCCESS in 1h 42m 44s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/5/check/devstack-plugin-ceph-dsvm-focal-py3/4178638/ : SUCCESS in 1h 51m 18s","accounts_in_message":[],"_revision_number":5},{"id":"3f8b00ab51cc7acdc8bfe03a64fec282d458179a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-20 00:54:29.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-5/ns5_nfs : SUCCESS in 3h 12m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-5/ns5_iscsi : SUCCESS in 2h 59m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"216e6a9bb6c02efeb4266e19970a069ccd56ea1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 07:07:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fc79961d1aa495b9322aff7b38ef807 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63138bc18e174b5e96ce1f47068c2003 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9471b7a9bd764735b60ef795fd0e7374 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e99790493fb54a9b936e0f400d793c2e : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7ccdee4aa534da3a3a4d839e4a4ced1 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc22ddd92f8446b906e0970b77f69c9 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/08bb4c1cf48a490dbd71a809f125b0de : SUCCESS in 1h 01m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/848fdec33d49440f925c47083c9766f2 : SUCCESS in 1h 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acef58d83aa74c66bacbe58afb93f947 : SUCCESS in 5m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6c679a836987415c9aa1e54be38cebd5 : SUCCESS in 10m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f662d06630d4ee9b400e1910737e8a6 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dda097e4a9ed46208455d67299c5d733 : SUCCESS in 5m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64bb2f12050f4300a4bad528ad2dd671 : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2981be226e1143dca90f92e75f5b8074 : SUCCESS in 43m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1695be08457c44bbac24dd79f0962504 : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c2f85db6652148cb9f4eb10f55adf960 : SUCCESS in 1h 12m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8e9da0d669e24771bd4a836ea3f55265 : FAILURE in 1h 17m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b53404fcde142d0b555005265f3e8f8 : SUCCESS in 1h 30m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a67474564d1d4c77b2c6d4289be5a256 : SUCCESS in 1h 26m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b65e8fcead4412dba4c03f9037fabe9 : SUCCESS in 1h 19m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/90cdc1ffc277486e82b92c5e5ada2669 : SUCCESS in 1h 19m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66b7d9f9ba2b4b53b66d2091bb08021f : SUCCESS in 2h 06m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f28b9bfe4fb43d28a411fbfde4dc9d2 : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":5},{"id":"cbd442591a90839098b073b69ef78e93e738b65a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-20 08:00:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/750782/5/upstream-solidfire-check/cinder-solidfire-iSCSI/1a4fe8e/ : FAILURE in 50m 43s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"111e7f0e8dc650f8de2b9db08733ecafe33462c4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-20 08:02:59.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/82/750782/5/check/DellEMC_PowerStore_iSCSI/7ae53c5/DellEMC_PowerStore_iSCSI/295 : SUCCESS in 54m 56s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/82/750782/5/check/DellEMC_PowerStore_FC/7464451/DellEMC_PowerStore_FC/289 : SUCCESS in 52m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"f0e580d7f1b8694adb7946c9a8ca2df2475332b8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-20 08:15:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/750782/5/upstream-check/cinder-cDOT-iSCSI/f6016a9/ : SUCCESS in 35m 32s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/82/750782/5/upstream-check/cinder-cDOT-iSCSI-IPv6/d0b528e/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/750782/5/upstream-check/cinder-cDOT-NFS/1e0856d/ : SUCCESS in 37m 44s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/82/750782/5/upstream-check/cinder-cDOT-NFS-IPv6/1219e03/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/750782/5/upstream-check/cinder-cDOT-FCP/40ca9d2/ : SUCCESS in 1h 06m 03s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"86dd324e21e24b4292fb2907ade4716db8416225","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-20 09:40:51.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/82/750782/5/check/DellEMC_PowerFlex/c3783e7/DellEMC_PowerFlex/251 : SUCCESS in 2h 32m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"9b496186822289cb4215b55435c8b4779d803cc9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-20 10:02:04.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/82/750782/5/check/DellEMC_Unity_FC/2aaad6a/DellEMC_Unity_FC/488 : SUCCESS in 1h 28m 25s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/82/750782/5/check/DellEMC_Unity_iSCSI/8a4ed88/DellEMC_Unity_iSCSI/453 : SUCCESS in 1h 51m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e0dc8c263a0075489c05ae613d091a7098d6fef0","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-01-20 10:19:21.000000000","message":"Patch Set 5:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/750782-5/Lenovo-iSCSI : SUCCESS in 52:01.97\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/750782-5/Lenovo-FC : SUCCESS in 45:19.88","accounts_in_message":[],"_revision_number":5},{"id":"f51ae4268baf310d92ac089255e90e473ad10c47","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-20 10:57:50.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/82/750782/5/check/DellEMC_SC_FC/a983926/DellEMC_SC_FC/240 : SUCCESS in 2h 15m 08s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/82/750782/5/check/DellEMC_SC_iSCSI/eb87884/DellEMC_SC_iSCSI/240 : ABORTED in 1h 25m 01s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"fefde913174675d2524051f6e51ccac72e129c61","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-20 12:47:01.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/82/750782/5/check/DellEMC_XtremIO_FC/9b6e185/DellEMC_XtremIO_FC/570 : FAILURE in 2h 05m 25s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/82/750782/5/check/DellEMC_XtremIO_iSCSI/7e97ac7/DellEMC_XtremIO_iSCSI/561 : SUCCESS in 1h 36m 45s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"53a309cc38cc489f9065550c4d161b4474787342","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-20 15:11:08.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/82/750782/5/check/DellEMC_VNX_iSCSI/b2341c0/DellEMC_VNX_iSCSI/290 : SUCCESS in 2h 48m 27s\n- DellEMC_VNX_FC http://publiclogs.emc.com/82/750782/5/check/DellEMC_VNX_FC/7d50cf0/DellEMC_VNX_FC/290 : SUCCESS in 1h 05m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d6c6aebd6fe2e7c8f83a33b33674dece264165e8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-20 18:53:42.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/5/20210120233436/NEC-ISCSI : SUCCESS in 2h 14m 06s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/5/20210120233436/NEC-FC : FAILURE in 4h 19m 03s","accounts_in_message":[],"_revision_number":5},{"id":"7ef088970846cc4151fc006d8899e5852954f12b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-07 14:21:16.000000000","message":"Patch Set 5:\n\nDo we have also any integration/scenario testing about this?","accounts_in_message":[],"_revision_number":5},{"id":"704e665350087775c40598135b3967212accf637","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-10 06:21:19.000000000","message":"Patch Set 5: Workflow-1\n\n\u003e Patch Set 5:\n\u003e \n\u003e Do we have also any integration/scenario testing about this?\n\nHi Luigi,\nNow that you mention it, we don\u0027t have a tempest scenario for this AFAIK. I should add one, and probably they should be merge together.","accounts_in_message":[],"_revision_number":5},{"id":"1d5d6ffc85d8a90b0981b4644b461e36ecb6623e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-11 10:17:16.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5: Workflow-1\n\u003e \n\u003e \u003e Patch Set 5:\n\u003e \u003e \n\u003e \u003e Do we have also any integration/scenario testing about this?\n\u003e \n\u003e Hi Luigi,\n\u003e Now that you mention it, we don\u0027t have a tempest scenario for this AFAIK. I should add one, and probably they should be merge together.\n\nWe need a multi-backend job and we don\u0027t have any right now. But I don\u0027t want to hold the fix... \n\nIt may be a bit complicated to automate this, or maybe not. It may be just a matter of:\n- creating a test which works if volume.backend_names is enabled and reverts a snapshot to all available backends\n- creating a job which defines two backends with devstack-plugin-ceph as primary/also used for backup.","accounts_in_message":[],"_revision_number":5},{"id":"d5ae77afc4abda3c2529806adec9a100934f2ac0","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-05 19:11:15.000000000","message":"Patch Set 5: -Workflow\n\n\u003e - creating a test which works if volume.backend_names is enabled and reverts a snapshot to all available backends\nI don\u0027t think it is necessary to add any new tests. We can use all the restore tests in tempest project. \n\n\u003e - creating a job which defines two backends with devstack-plugin-ceph as primary/also used for backup.\nhttps://review.opendev.org/c/openstack/cinder/+/784815","accounts_in_message":[],"_revision_number":5},{"id":"f36c90e86c215971fb8280f38dbab978fd8fe579","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-04-12 13:44:36.000000000","message":"Patch Set 5:\n\nLooks like the revert test passed:\n```\ncinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/e2fe565732374f2fbb8cde7b7b550a7a : SUCCESS in 1h 23m 11s (non-voting)\n```","accounts_in_message":[],"_revision_number":5},{"id":"fb6fd994e86e676438d62cf87c872496a6a58d79","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-12 14:31:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"afbe3e17c45b20f8cfcc500fdf3809051c5e6bd9","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-04-20 19:01:18.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d9caa2919a39bbbf69ea6eec52867eb353708b30","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-08-10 06:28:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c9f81b68ea4531ce9d39b5373693ea1775481de9","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-08-18 19:23:19.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ff9f26d632a1ac1968ef7d05621806ab16513a9e","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-08-30 18:56:32.000000000","message":"Patch Set 5: Code-Review-1\n\n(11 comments)\n\nThank you Sofia for working on this issue!","accounts_in_message":[],"_revision_number":5},{"id":"464d97bd021d9669df5599d1ae833ff7ef346c58","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-31 14:22:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"e03e2d2c81a30fa680ded12ef8d2e869531de93b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-09-02 16:40:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9093f68ea1048d84bd1b6dfeeaa5772d43e97506","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-09-02 16:40:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"5a3bedd1d8424eb4b3ceb78111ab909bbbf69527","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-09-02 16:51:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"e146cc6ba6cd7221d3ad853eac680080e44ca1ef","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-09-02 16:55:15.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"ad07f06e0e009224718c545ab6eeeb57a003797e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-09-02 16:55:47.000000000","message":"Patch Set 7: Workflow-1\n\nNeeded to rebase the patch before going on with the fixes","accounts_in_message":[],"_revision_number":7},{"id":"a89bb1ff69f795aec1f43377bf0cb80c0aa79f52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 17:07:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6911b74c54b2401da37b4617498aa41c : FAILURE in 13m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c211713e066d46858c6f54a33962b611 : FAILURE in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"46c5468c0df05c3542d886c7ce2e6f7c17dab613","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-09-02 17:33:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"696c776dbe0a903c1e6206615df6e7540816fbd5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-09-02 17:53:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"888a380d3cb93a8bea2ce52beb0ba177a144360e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 18:00:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8ce33173fda484c952ee159f43332c0 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ce894f136d04096b119328824477e2d : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"974ed1933f921b8c000cbf23f4866fca7779c0b0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-02 18:31:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/8/iscsi/ : SUCCESS in 56m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/8/smb/ : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":8},{"id":"6d525487d586e9f20aa332d2b30c4d6633da5f49","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-02 19:21:50.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/82/750782/8/check/iscsi-driver-primera/b8c9e35/b8c9e35.html : FAILURE in 34m 34s","accounts_in_message":[],"_revision_number":8},{"id":"6d200c38ecc74f75bc32d39c20e1ab41c82a4f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 19:50:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d12a75ac91ec472aadd7bfb8219ce4dc : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c697ba17fe5492ba2c33e151bfc5751 : SUCCESS in 11m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b50418e65c4c45c0a9a5ba0f4c71e0bb : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49558b2542864e7b9181d1d59d008261 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ee54ced20bd4edd868ef1679562ac15 : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/00c7dc8eb238442287e8870bd633cfba : SUCCESS in 1h 04m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e79158bdb5644578d68de4e6b848dc7 : SUCCESS in 1h 04m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6982a8b35f4344b5bb73f4faa56688d0 : SUCCESS in 7m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5cc3ce5f4c974563b756fd0c222191c9 : SUCCESS in 13m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b1bd5972321741f08d30082b0b7392dd : SUCCESS in 6m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cef1c776ea5245c4ae3f44e63aafab94 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf4e38914c464796b99a8cef10c87852 : SUCCESS in 8m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fa15251842bc41228c69b58d02a19bce : SUCCESS in 8m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c82e06b5420e486fb3a6dc35ecf83815 : SUCCESS in 38m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b55ba78a69fd4decb5c1798212492caa : FAILURE in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b37ad98e11404f13948a720dad5a717a : SUCCESS in 1h 22m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3e9a588cb43949419423e3a057b2318c : SUCCESS in 1h 41m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e581ea7e0f714a56b86541d2beee1fa6 : SUCCESS in 1h 35m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9ab5880999d44780bba9445882d1104f : SUCCESS in 1h 10m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/291ea8ac424c4b1f9dc1f01df9796efa : SUCCESS in 1h 12m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cf918628856d4246ac9ace9863888792 : SUCCESS in 1h 09m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e34b12477c84d888609f67fd2bc32e0 : SUCCESS in 2h 09m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e9dc0d9f1d940acb48a7e49a2f6b099 : SUCCESS in 46m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a33df15926248228553ca4030ea084a : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":8},{"id":"10322f21760553a161732a924b5e420940054c60","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-02 21:02:30.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5e78ff4/ : SUCCESS in 2h 20m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cc572e7/ : SUCCESS in 1h 40m 05s","accounts_in_message":[],"_revision_number":8},{"id":"5b9977ac0794237dd3ec3865eb86c7fc378d1a45","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-02 21:04:27.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- iscsi-driver-alletra9000 https://hpe-storage.github.io/hpe_cinder_logs/logs/82/750782/8/check/iscsi-driver-alletra9000/b911038/b911038.html : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":8},{"id":"d61ec83dfdbfcb64ede4aa615ee8b540ebcfde47","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-02 21:27:30.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/750782/8/upstream-check/cinder-cDOT-iSCSI/b0f3fe4/ : SUCCESS in 34m 46s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/750782/8/upstream-check/cinder-cDOT-NFS/19691e6/ : SUCCESS in 35m 10s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/750782/8/upstream-check/cinder-cDOT-FCP/ced21a7/ : SUCCESS in 1h 12m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"fad9e834737ef74263e31094eafc4f82107c89c5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-09-02 21:33:44.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/750782/8/upstream-solidfire-check/cinder-solidfire-iSCSI/1751c3e/ : FAILURE in 1h 41m 55s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"9e53a78fcdb018bb60c7d5ef23cb166febf8e08e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-03 00:50:56.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/8/check/iscsi-driver-nimble/a5561b7/a5561b7.html : FAILURE in 1h 25m 54s","accounts_in_message":[],"_revision_number":8},{"id":"b7978f99d40f36c182947b34104ff324f0e27337","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-09-03 03:34:48.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/8/check/DellEMC_PowerStore_iSCSI/ddd5a92/DellEMC_PowerStore_iSCSI/475 : SUCCESS in 1h 28m 38s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/750782/8/check/DellEMC_PowerStore_FC/15edd97/DellEMC_PowerStore_FC/507 : SUCCESS in 1h 10m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"e002f530fc597f13655d0824d8790f86a17352e8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-09-03 06:35:20.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/750782/8/check/DellEMC_SC_FC/c536b97/DellEMC_SC_FC/408 : SUCCESS in 47m 20s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/8/check/DellEMC_SC_iSCSI/f995ae9/DellEMC_SC_iSCSI/386 : SUCCESS in 1h 40m 58s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"d7ebbfdbe26a04499aadc083a3c34d0983588fd4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-03 12:10:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/750782/8/check/dsvm-tempest-cisco-zonemanager-job/49368d4/ : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":8},{"id":"48a980b22a8ca352b33c54087940b564711b32e9","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-09-03 13:43:53.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/82/750782/8/check/DellEMC_PowerMAX_FC/07d22fe/DellEMC_PowerMAX_FC/484 : FAILURE in 2h 56m 22s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/8/check/DellEMC_PowerMAX_iSCSI/8b0fd2b/DellEMC_PowerMAX_iSCSI/512 : FAILURE in 2h 39m 27s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"5bcbb343b9958a94597fcb3e7ae557e1db559623","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-04 15:57:29.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-6/ns5_nfs : FAILURE in 1h 21m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-6/ns5_iscsi : FAILURE in 1h 21m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"fce5a80841abbaeb4a8304b74f1a10bf8d97d22c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-04 20:48:56.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-7/ns5_nfs : SUCCESS in 4h 43m 25s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-7/ns5_iscsi : SUCCESS in 4h 34m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"cb78df83b9c44ec0505f09c5e38be11d97c1a066","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-04 21:22:58.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-8/ns5_nfs : SUCCESS in 4h 22m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-8/ns5_iscsi : SUCCESS in 4h 31m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"0c5b4d1e7d1212626408ac7a7535bcfb291aed6b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-09-05 06:11:01.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/750782-8/HPMSA-iSCSI : SUCCESS in 1:00:41\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/750782-8/HPMSA-FC : SUCCESS in 34:20.10","accounts_in_message":[],"_revision_number":8},{"id":"a7af4fdb0635336339836b0cc7c12700b51bb485","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-05 15:28:24.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/750782/750782/8/check-not-vote/ext-cinder-zuul/f89b23e : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":8},{"id":"c7c727c2789eb69f362236b6fbb3f708df29e9ab","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-09-05 22:35:32.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/750782-8/Seagate-iSCSI : SUCCESS in 1:13:07\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/750782-8/Seagate-FC : SUCCESS in 40:07.53","accounts_in_message":[],"_revision_number":8},{"id":"be10967ce785f8d58d881805d8cadf1025743bb0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-06 08:43:43.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-8/ns5_nfs : FAILURE in 53m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-8/ns5_iscsi : SUCCESS in 4h 27m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"04a990fbf924df2f54ebd92a4da22fe0aa277a76","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-09-08 18:19:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/8/20210909002526/NEC-ISCSI : SUCCESS in 2h 47m 46s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/8/20210909002526/NEC-FC : SUCCESS in 2h 54m 03s","accounts_in_message":[],"_revision_number":8},{"id":"3d4afdb6215c174cb72fcd5081e5bbfde571a677","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-09-09 03:26:45.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/750782-8/Lenovo-iSCSI : SUCCESS in 1:22:31\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/750782-8/Lenovo-FC : SUCCESS in 51:46.78","accounts_in_message":[],"_revision_number":8},{"id":"3dabc554f835a2cea240b54898e96fdbdd4c2caf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-11-12 20:21:53.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"e21976f4a133271214d61e5f68e37f227a95a58e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 20:51:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eafe6960ef00472aa2bba017446c2c4a : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27add01da98546b3958092e322bf3136 : SUCCESS in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5c7caf89106c88a6fef032376eb767b0a7dc44ad","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-12 21:32:43.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/9/iscsi/ : SUCCESS in 1h 10m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/9/smb/ : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":9},{"id":"dee690f82fb687c5318702b6e9dc8e437d2fe363","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-12 21:44:29.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_Unity_FC/3fbed6e/DellEMC_Unity_FC/162 : SUCCESS in 1h 05m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_Unity_iSCSI/8422e96/DellEMC_Unity_iSCSI/163 : SUCCESS in 21m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"a0860f4ad007a8dc37c0acc39f842cfcfeb2ea13","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-12 21:51:14.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/750782/9/check/dsvm-tempest-cisco-zonemanager-job/a4a0003/ : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":9},{"id":"b3220c1d7fd25275eb8701af7f1faa510f6b524b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-12 21:58:41.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/82/750782/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c27778d/c27778d.html : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":9},{"id":"fabd25d69423728055c10ccc1a37589b164b40e7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-12 22:00:00.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f155b9d/ : SUCCESS in 1h 36m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3c041c8/ : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":9},{"id":"e83c6bb7517490b7c9709fe74d1d0979ab947a17","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-12 22:00:48.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/82/750782/9/check/iscsi-driver-primera/f5a4a48/f5a4a48.html : SUCCESS in 1h 17m 39s","accounts_in_message":[],"_revision_number":9},{"id":"7615b9126957a5048955c8930b706e7c58a1173f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-12 22:07:53.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/9/check/iscsi-driver-alletra-6k/2cdb8d3/2cdb8d3.html : SUCCESS in 1h 40m 25s","accounts_in_message":[],"_revision_number":9},{"id":"a72fef570bf46fecedf49e80de01a8cc27b08515","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-12 22:13:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ab053f5f37844d96ace2f68be78fee40 : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":9},{"id":"050b8ae033d41d7baff6012d172daba62be652dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 22:35:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982cda7dc2e0497bbd5ece5855885979 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e05443342c04b0d9b0310e0d23e76ff : SUCCESS in 11m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16b6be0d0abb4ce69c4f75578764db63 : SUCCESS in 13m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c2e174532f14cc4aeb8a7cbd1be36ec : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f655628c1381464492e9f04461f6acb2 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d41cb77153714dd496a332897ef599e7 : SUCCESS in 59m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c0b659875154d8fbb23be71da5d2a70 : SUCCESS in 1h 24m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93838d3f63745098d96314c40bacdd1 : SUCCESS in 8m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ecabb2a22e604a08a70dba995ee09fc2 : SUCCESS in 16m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c8c88b678df248b18bea78661197a191 : SUCCESS in 6m 22s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d1c67ffd4264a4ba00862ea20dbe6ae : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7f8f0ea7c994e5dbe4e59dc19148722 : SUCCESS in 7m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61c1d989b68f492386cd99f60c7552f7 : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/74cb5b1f50584258b4583807045d84ce : FAILURE in 28m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa432c3c649c4cd9ada884ba59557132 : FAILURE in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f02224ee7d543a3921d954e7032a796 : SUCCESS in 1h 36m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8023166960b745bca0fbdbc28b1dad45 : SUCCESS in 1h 27m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f77023a594e4009882a1af1d79d7793 : SUCCESS in 1h 58m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8ca8cf06b0c4519bb23f2b92455fa2a : SUCCESS in 59m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/063877b1722c463a8f9ec25e07961827 : SUCCESS in 1h 21m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca1fd9e6904740d0ad43726e41834368 : SUCCESS in 55m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3d3306f9d9c341b9a02b1330b82077a2 : SUCCESS in 1h 32m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4c29a16668ca46bca60e5c6a94cbc295 : SUCCESS in 2h 08m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b60ca3d939cd44508fbd408adb47dfc3 : SUCCESS in 50m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/348fef1933344d8ba44044cfa13ab82b : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":9},{"id":"571fcdded8e651d1bfb8d671dab94334f55d5e80","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-12 22:59:37.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/9/check/devstack-plugin-ceph-dsvm-focal-py3/66a137b/ : SUCCESS in 2h 21m 41s","accounts_in_message":[],"_revision_number":9},{"id":"7bfa77dfbb6f12dd36ea52e2bf0b683078bcbaab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-12 23:07:20.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/9/check/iscsi-driver-nimble/d314676/d314676.html : SUCCESS in 1h 49m 22s","accounts_in_message":[],"_revision_number":9},{"id":"84e23ddb13fc754bc338fd43ed7f4c7aa5c8c601","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-12 23:11:17.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/750782/9/upstream-check/cinder-cDOT-iSCSI/7d9425e/ : SUCCESS in 34m 08s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/750782/9/upstream-check/cinder-cDOT-NFS/2af0321/ : SUCCESS in 33m 48s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/750782/9/upstream-check/cinder-cDOT-FCP/0a2dfb2/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"39611f6d8fa635db16b7929ade61a112af969150","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-12 23:41:10.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/750782/9/upstream-solidfire-check/cinder-solidfire-iSCSI/70f9f1d/ : SUCCESS in 1h 03m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"361dcc459b08b0377a49f27ccdf73d8eed2e5c53","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-13 00:07:06.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_XtremIO_FC/0ed51bf/DellEMC_XtremIO_FC/125 : SUCCESS in 55m 32s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_XtremIO_iSCSI/70061b8/DellEMC_XtremIO_iSCSI/151 : SUCCESS in 1h 54m 26s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"1bf2e10b90c2fbe2e8631c0e54b23981ecc9e002","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-13 01:11:17.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_PowerFlex/c7febfb/DellEMC_PowerFlex/83 : SUCCESS in 1h 45m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"487e52659961b0ae2567dea75104ebc38b236659","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-13 01:15:40.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_Unity_FC/d95c498/DellEMC_Unity_FC/165 : SUCCESS in 1h 13m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_Unity_iSCSI/399c522/DellEMC_Unity_iSCSI/166 : SUCCESS in 23m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"ebb1974ecbbd05f7b318044548535c7f538c5f07","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-13 01:23:01.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_PowerStore_iSCSI/a130546/DellEMC_PowerStore_iSCSI/61 : SUCCESS in 1h 52m 09s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/750782/9/check/DellEMC_PowerStore_FC/63e0aa8/DellEMC_PowerStore_FC/68 : SUCCESS in 1h 20m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"427cc25a2c8b9b9711c94a0b17e90d9092d21b6d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-13 02:12:12.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-9/ns5_nfs : SUCCESS in 4h 15m 36s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-9/ns5_iscsi : SUCCESS in 4h 21m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"7d5eaa63e8bf2ec45a31ecc533b1e1f0c8b93427","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-13 02:37:57.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/750782-9/HPMSA-iSCSI : SUCCESS in 1:11:09\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/750782-9/HPMSA-FC : SUCCESS in 44:55.85","accounts_in_message":[],"_revision_number":9},{"id":"5efe8a06ba5d7007f49708d0047258b24bb088ad","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-11-13 02:55:00.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/750782-9/PVME-iSCSI : SUCCESS in 1:00:21\n- PVME-FC http://os-logs.tristero.net/dell-ci/750782-9/PVME-FC : SUCCESS in 1:00:25","accounts_in_message":[],"_revision_number":9},{"id":"3cf2915c1c7ecdd046d8fc034b6baa8f9dd0cd61","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-11-13 03:35:20.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/750782-9/Seagate-iSCSI : SUCCESS in 1:18:57\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/750782-9/Seagate-FC : SUCCESS in 49:56.33","accounts_in_message":[],"_revision_number":9},{"id":"b62a2be29970412bc7042213f671bf89a6234cd9","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-13 05:09:57.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/750782-9/Lenovo-iSCSI : SUCCESS in 1:27:39\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/750782-9/Lenovo-FC : SUCCESS in 57:38.98","accounts_in_message":[],"_revision_number":9},{"id":"a1727dd7c461ccc27a9c207f928fca9d719dd278","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-13 15:45:06.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/750782/750782/9/check-not-vote/ext-cinder-zuul/d1aa099 : SUCCESS in 31m 04s","accounts_in_message":[],"_revision_number":9},{"id":"7f443cec2835095ebee12be884224933da4798bc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-13 22:31:38.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/9/20211114043746/NEC-FC : SUCCESS in 2h 52m 15s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/9/20211114043746/NEC-ISCSI : SUCCESS in 2h 53m 49s","accounts_in_message":[],"_revision_number":9},{"id":"bec58ded608e9f6f3250ba1c719da4ecd0c7e5a9","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-19 15:05:58.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/82/750782/9/check/dsvm-tempest-TOYOU-ACS5000-driver/6978e13 : FAILURE in 19m 33s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/82/750782/9/check/fc-test/a28b7a0 : SUCCESS in 1h 42m 37s","accounts_in_message":[],"_revision_number":9},{"id":"ab2c59ceea494584a3b2bb173b9444958d92c158","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-01-10 20:22:20.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"6c3731eb60d2e314f367bfbac11e6810533a0021","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-10 20:22:32.000000000","message":"Patch Set 10:\n\nBuild started","accounts_in_message":[],"_revision_number":10},{"id":"4b71344120b00613eabd703d0834a461e457edb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-01-10 20:23:10.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"b2f70bfd8b4893189d199ec0c9169f9d1ac93b8b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-10 20:25:48.000000000","message":"Patch Set 10:\n\nBuild SUCCESS. log path: https://raw.githubusercontent.com/LightBitsLabs/logs/main/10_01_2022_20:25:46/run.log","accounts_in_message":[],"_revision_number":10},{"id":"df686ac3475a57b0eaa791fc3310b28a329ecb84","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-10 20:25:49.000000000","message":"Patch Set 11:\n\nBuild started","accounts_in_message":[],"_revision_number":11},{"id":"d579643f7ca9d4991e2dba346fc023ce627f8537","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-10 20:30:17.000000000","message":"Patch Set 11:\n\nBuild SUCCESS. log path: https://raw.githubusercontent.com/LightBitsLabs/logs/main/10_01_2022_20:30:15/run.log","accounts_in_message":[],"_revision_number":11},{"id":"5098fa064ccf3fd2e69f3070dd4979161b7b9428","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 20:46:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df7b8ae5b27742b1a5bd66d867c6896e : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2976838a472b44dab448fdb4d470c314 : FAILURE in 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c2540df1dfcdbee7b18be02fbdec2a2300abaa23","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-10 21:36:32.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/11/iscsi/ : SUCCESS in 1h 12m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/11/smb/ : SUCCESS in 1h 10m 25s","accounts_in_message":[],"_revision_number":11},{"id":"d7b4bc0170774c9193a07d11f38295265d38a3b8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-10 22:06:21.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/11/ : SUCCESS in 1h 33m 50s","accounts_in_message":[],"_revision_number":11},{"id":"fa804c8eadbd7562bab696bd98e36b27adedebe0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-10 22:15:59.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/750782/750782/11/check-not-vote/ext-cinder-zuul/b51560a : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":11},{"id":"859a4097741072ed1227e29e88d77127b4de13ac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-10 22:40:38.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_Unity_FC/2292489/DellEMC_Unity_FC/356 : FAILURE in 1h 17m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_Unity_iSCSI/2e7281f/DellEMC_Unity_iSCSI/349 : SUCCESS in 1h 17m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"652ed1b8309461809ef47a02faab091567b59ef1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-10 22:49:03.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/11/check/tempest-dsvm-full-focal-py3/fb9acc7/ : SUCCESS in 1h 58m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/11/check/devstack-plugin-ceph-dsvm-focal-py3/d32e226/ : SUCCESS in 2h 20m 50s","accounts_in_message":[],"_revision_number":11},{"id":"3af3c40de3fb0b55fb855c4a0ed8374dd39d7ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 23:21:01.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d4bbda5f5546ed8e70c48bdb89b84d : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a26ca3593394a3eb2213e2aa7670585 : SUCCESS in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c10aaca831444059beb2ae679a716deb : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e5a35402584655aba7b3689e8b5946 : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/10da3690550f4ad6958c88b6b276cb80 : SUCCESS in 57m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a14eaa83421741d0a452fd7401ce054d : SUCCESS in 1h 33m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/754814faede641ac91f071594e0e7d49 : SUCCESS in 46m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41b7096165054d3884c0de5a7f5c12c3 : SUCCESS in 6m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bc1881f557024fd9ab56e0f3be7b4b17 : SUCCESS in 14m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cb865e84ed64448091c078f662f0a55c : FAILURE in 6m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6ef6d5ffc3d4c4cad3a0809cb6df680 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee8bc2f3338f49928b321ec338655ae6 : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48a0b2ec3b6345d4b01d40f830b44d6d : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9b200b5e88934ce08a1ff85f40fe13ae : SUCCESS in 37m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b83bdf8b9a141398e5f8aaf0e7bb533 : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0260571e7c7247f5830e8ca0d7d8911b : SUCCESS in 1h 34m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e263c0c2fd514deda506c8e9cb06f933 : SUCCESS in 1h 26m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7fce125e5080427da6cd35eed1620b1c : SUCCESS in 1h 31m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e0a45e92c73488080cc69cdef5b0572 : SUCCESS in 1h 15m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/07fa30a2a5c140a589e2b03539dfadd7 : SUCCESS in 1h 35m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e9e06fc67ecb4540b1058774bbd214f0 : SUCCESS in 43m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5820462d7cb2433bb3fe420a4c6c2062 : SUCCESS in 1h 11m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7e3f883a4784f259da6f521ba332e57 : SUCCESS in 2h 49m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4c849f8f10e4623abd93f929b9107ff : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":11},{"id":"6f86a4a1788664ffdeee5741ca9b74470edd30c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-10 23:22:40.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/15450bc15bca403898cd7118e801e3e8 : FAILURE in 1h 28m 00s","accounts_in_message":[],"_revision_number":11},{"id":"70825597b7725cc679c46599f70e4d92ae6ee4f4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-10 23:44:13.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_XtremIO_FC/4b6bbc7/DellEMC_XtremIO_FC/434 : SUCCESS in 56m 45s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_XtremIO_iSCSI/4ee268a/DellEMC_XtremIO_iSCSI/447 : SUCCESS in 1h 18m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":11},{"id":"d1ce47278b6fec61b00b91dcf48e727c73190fe3","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-10 23:49:13.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/750782/11/upstream-solidfire-check/cinder-solidfire-iSCSI/bf1d658/ : FAILURE in 26m 58s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":11},{"id":"887b8cb583cf3b29e6785ced0b6e2ef14594382e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-11 00:34:32.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_Unity_FC/5949012/DellEMC_Unity_FC/357 : FAILURE in 1h 13m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_Unity_iSCSI/f3ab157/DellEMC_Unity_iSCSI/350 : SUCCESS in 1h 10m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"fffdfb550ea6a02f7459b05001a8ae65868e5027","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-11 00:37:45.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/750782/11/upstream-check/cinder-cDOT-iSCSI/5555da6/ : FAILURE in 7m 47s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/750782/11/upstream-check/cinder-cDOT-NFS/66bbc3b/ : SUCCESS in 40m 19s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/750782/11/upstream-check/cinder-cDOT-FCP/86618b6/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"a1e495873b6b4b97bedd5b8ee2b85d7216496aee","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-01-11 00:43:17.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_PowerStore_iSCSI/0711394/DellEMC_PowerStore_iSCSI/217 : SUCCESS in 1h 21m 57s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_PowerStore_FC/4d309dc/DellEMC_PowerStore_FC/221 : SUCCESS in 56m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"a7ca69a6e029795b271c5d24e9c7284e854b09bc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-11 00:49:32.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_PowerFlex/5264f83/DellEMC_PowerFlex/247 : SUCCESS in 1h 28m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"248ee5c2fcc7c18eb66904bf76d1ab78aaa761d7","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-01-11 00:53:39.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_SC_FC/cf30976/DellEMC_SC_FC/185 : SUCCESS in 41m 40s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_SC_iSCSI/89151be/DellEMC_SC_iSCSI/182 : SUCCESS in 1h 32m 23s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":11},{"id":"3bc3e00960db9a2674ecb44c4578700b63617f7d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-11 00:57:42.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-10/ns5_nfs : SUCCESS in 4h 34m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-10/ns5_iscsi : SUCCESS in 4h 19m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"572fab4849aea52fb33a4c94294ba0c68b94a8ed","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-11 01:24:41.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-11/ns5_nfs : SUCCESS in 4h 10m 12s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-750782-11/ns5_iscsi : SUCCESS in 4h 16m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"4672faf68c4b8e595a17db64f542e5ab15c9af2c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-11 01:41:14.000000000","message":"Patch Set 11:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":11},{"id":"a1954ee88c8042b1860e08e8e0a80d0384d0ecdd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-11 01:45:54.000000000","message":"Patch Set 10:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":10},{"id":"c9eab400e43cdbf2e6f2588afdd08c63186c3533","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2022-01-11 04:06:47.000000000","message":"Patch Set 11:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/82/750782/11/check/dsvm-tempest-TOYOU-ACS5000-driver/8f66dcc : SUCCESS in 2h 00m 42s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/82/750782/11/check/fc-test/648b6fd : SUCCESS in 2h 02m 02s","accounts_in_message":[],"_revision_number":11},{"id":"fa2fe9c7efd4729e258667eb4472f728867b2026","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-01-11 04:28:08.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/750782-11/PVME-iSCSI/ : SUCCESS in 56:34.14\n- PVME-FC https://os-logs.seagate.com/dell-ci/750782-11/PVME-FC/ : SUCCESS in 59:08.12","accounts_in_message":[],"_revision_number":11},{"id":"23d0b8341c1d548f866cf7461959d92304531cfb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-11 07:02:14.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/3990c93509ac41b28259d3867bcee7d6 : RETRY_LIMIT in 3h 08m 24s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/01e78763810d49e99d9ac229ed2816ab : RETRY_LIMIT in 1h 41m 57s","accounts_in_message":[],"_revision_number":11},{"id":"c26e8ccfddf1885b6ca1105c0632db279cfee6f3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-01-11 13:21:05.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/750782-11/HPMSA-iSCSI/ : SUCCESS in 50:02.52\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/750782-11/HPMSA-FC/ : SUCCESS in 43:03.56","accounts_in_message":[],"_revision_number":11},{"id":"c7fa5830f9da9b1fe6b94e6133ba3e42ada6fe00","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-01-11 14:53:15.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/750782-11/Lenovo-iSCSI/ : SUCCESS in 1:01:27\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/750782-11/Lenovo-FC/ : SUCCESS in 56:54.74","accounts_in_message":[],"_revision_number":11},{"id":"2d28a99890d1f34a18ae1b78ef2b543421bd8cb2","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-01-11 17:07:17.000000000","message":"Patch Set 11:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/750782-11/Seagate-iSCSI/ : FAILURE in 11:49.99\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/750782-11/Seagate-FC/ : FAILURE in 11:38.90","accounts_in_message":[],"_revision_number":11},{"id":"3ba41c42263d7c44c5575b4a4092cd44d40dd923","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-12 02:18:15.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/11/20220112052730/NEC-FC : SUCCESS in 2h 52m 21s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/11/20220112052730/NEC-ISCSI : SUCCESS in 5h 50m 41s","accounts_in_message":[],"_revision_number":11},{"id":"7f2fca5a272b1c8890eda3061878e362f0fef021","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-17 01:18:46.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_PowerMAX_FC/ef2c6ab/DellEMC_PowerMAX_FC/168 : FAILURE in 1h 27m 17s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/11/check/DellEMC_PowerMAX_iSCSI/d6dc52c/DellEMC_PowerMAX_iSCSI/177 : FAILURE in 2h 51m 29s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":11},{"id":"3ae82ff431fee5ac2c77ea42fcb24cf6f7b1db12","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-10 22:29:47.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1\n\n(14 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d1a518a58154a12df546bf8c39a0ef1c126691f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-11 15:23:01.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"fe184f1e08ec731368df537a03a703dc977eff87","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-11 15:23:18.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d97741ea2ce08bbe666d519bc0875702e788e718","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-11 16:07:32.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/82/750782/12/check/iscsi-driver-primera/60462e8/60462e8.html : FAILURE in 43m 02s","accounts_in_message":[],"_revision_number":12},{"id":"db6075e26d75276403d65b00441a33e72d3fd974","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-11 16:07:49.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/82/750782/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c81fae7/c81fae7.html : FAILURE in 43m 18s","accounts_in_message":[],"_revision_number":12},{"id":"fa0b3d633609d55cdac706e129d3c42e0ad7f229","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 16:22:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/631ad3483a714f2c944b2eb3ce3a94a5 : FAILURE in 25m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc1b2bbf796744c79ea1d55e8c3d94ae : SUCCESS in 30m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ae803448af7175da74b5db47c1ba5c980b14a7b8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-11 16:25:30.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/12/check/iscsi-driver-alletra-6k/d98fc4a/d98fc4a.html : FAILURE in 1h 00m 46s","accounts_in_message":[],"_revision_number":12},{"id":"ad5644d604792c7981baa4614e4fc30d3f4f3264","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-11 16:44:07.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/12/iscsi/ : SUCCESS in 1h 20m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/12/smb/ : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":12},{"id":"10150aea6a81c115c45eab5370e23ed14838e3a2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-11 17:13:33.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cb2a1e3/ : SUCCESS in 1h 30m 40s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/abbed25/ : FAILURE in 1h 48m 19s","accounts_in_message":[],"_revision_number":12},{"id":"f00662aef876e9233028ba947344a371ea96f260","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-11 18:51:47.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/615d851b096647c0844bc7e568473ae1 : POST_FAILURE in 2h 25m 04s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/dcb15cbaa8dd418ab57fb7606c49894c : TIMED_OUT in 3h 12m 45s","accounts_in_message":[],"_revision_number":12},{"id":"b154bcd0526398dec3055f49949040a54befc64b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 18:58:11.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b05fa7ddbb4da59e08167b2acf5f1f : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08974c16db784560b4b3704f3d5d58c2 : SUCCESS in 14m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd1e481a414c432294b239147a4cd2d8 : SUCCESS in 16m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548ada671da9483ebd2f39698bb013a0 : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/328d13e4e958426a8159cb3a22475236 : SUCCESS in 48m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ac558452a68e4f89afca709f947c1336 : SUCCESS in 1h 08m 53s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19b24edff03148849f741e63a169c0b2 : SUCCESS in 1h 09m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88735182566e4a9cb35bffbe3ddfc567 : SUCCESS in 53m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8312acabd5e64b128bf7a8b1adc0438a : SUCCESS in 9m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1e7a0a174be24aeea884da2560a653a2 : SUCCESS in 17m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d32265a39a764ac5ac7e993e36dbc40a : SUCCESS in 9m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ca267df66ab4520916e20e4c3472839 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86da9de7ecf740de8164796d59dc5895 : SUCCESS in 9m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28613db1657146c5a86e5a7066602a89 : SUCCESS in 10m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6ee2cba7bb1c4fc898584885048452b6 : SUCCESS in 43m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/943a94c70c44462d9625bf0e9661ea0f : FAILURE in 9m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6a62e2b079ad4443857cffbd3868d343 : SUCCESS in 1h 19m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/36aeed628af4442aa5e529cd6ab32b2e : SUCCESS in 1h 16m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/21a08bc6ac76480180382a4f3fc6e05d : SUCCESS in 1h 42m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b062b79a18aa431eaf160079e974f53d : SUCCESS in 1h 05m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/181bba74b07e47ed9e429c5bcf59598e : SUCCESS in 1h 19m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76929e63eb3a48dfaaf2be7d74e9d066 : SUCCESS in 42m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb97fb6a077d4f309d66da8a499024de : SUCCESS in 1h 40m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bcfcf3429a4a4f9aa40dd58eb8bfe29b : SUCCESS in 2h 24m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/366ac6e83e024367b05e21218e85b3ed : SUCCESS in 51m 29s","accounts_in_message":[],"_revision_number":12},{"id":"5c61203792cdd74af3b86625e174162caaf5baed","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-11 19:58:17.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://93.100.254.243/nexenta/refs-changes-82-750782-12/ns5_nfs : SUCCESS in 4h 30m 4s\n* NexentaStor5_Cinder_iSCSI http://93.100.254.243/nexenta/refs-changes-82-750782-12/ns5_iscsi : SUCCESS in 4h 34m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"10050785732e46091cbb8732fee61646da02740b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-03-11 22:00:06.000000000","message":"Patch Set 12:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/82/750782/12/fc : SUCCESS in 1h 04m 22s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/82/750782/12/iscsi : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":12},{"id":"e01ef7910627e76607815a5ceeb41aa5748d2e0a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-12 02:56:16.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job finger://zuulserver:17979/b94ef75660da4214906e761e022dbc69 : RETRY_LIMIT in 1h 06m 38s","accounts_in_message":[],"_revision_number":12},{"id":"4bb3df21c0a139092effaeac506870e087673349","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2022-03-12 02:58:10.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"0865ad0d684803e74de4d35b2a84f6c0c3588509","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-12 04:15:21.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/750782/750782/12/check-not-vote/ext-cinder-zuul/71d5e85 : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":12},{"id":"cde910727edf187ae07a81b0602882b5372d9d28","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-12 08:52:28.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/12/ : SUCCESS in 1h 55m 46s","accounts_in_message":[],"_revision_number":12},{"id":"a1a31d3c0d3b43cfabb876962cb1a6e0c0bee309","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-03-12 23:09:03.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/750782-12/HPMSA-iSCSI/ : SUCCESS in 50:59.26\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/750782-12/HPMSA-FC/ : SUCCESS in 44:58.25","accounts_in_message":[],"_revision_number":12},{"id":"9d2e236215b9f12b0acfa47dd54bcae1ee70148e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-13 01:18:14.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/12/check/iscsi-driver-nimble/68c614b/68c614b.html : FAILURE in 1h 00m 26s","accounts_in_message":[],"_revision_number":12},{"id":"f856541790b7033d93af16780936228ea14cd919","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-13 02:19:40.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/12/check/iscsi-driver-nimble/4a08b76/4a08b76.html : FAILURE in 59m 42s","accounts_in_message":[],"_revision_number":12},{"id":"40749e6617b4f88df8b67f1ea291056b5b25ae98","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-03-13 23:37:57.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/750782-12/Lenovo-iSCSI/ : SUCCESS in 1:03:15\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/750782-12/Lenovo-FC/ : SUCCESS in 59:41.26","accounts_in_message":[],"_revision_number":12},{"id":"d3f7b960b45ae63473d73ab6dca2244b40d075c1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-03-14 04:03:04.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/750782-12/PVME-iSCSI/ : SUCCESS in 1:07:31\n- PVME-FC https://os-logs.seagate.com/dell-ci/750782-12/PVME-FC/ : SUCCESS in 1:04:33","accounts_in_message":[],"_revision_number":12},{"id":"48dc32a11cfe43ae51ce48bd0d458932d2dd2447","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-03-14 09:38:13.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/750782-12/Seagate-iSCSI/ : SUCCESS in 1:12:18\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/750782-12/Seagate-FC/ : SUCCESS in 1:05:18","accounts_in_message":[],"_revision_number":12},{"id":"61d2c74c91db44a3a82d322c73a8785c964bcdf9","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-03-14 09:41:20.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"e3709ae3b465a917f62c658262a083bd1098dd95","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-03-18 00:12:20.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/12/20220318055137/NEC-FC : SUCCESS in 3h 17m 23s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/12/20220318055137/NEC-ISCSI : SUCCESS in 3h 20m 40s","accounts_in_message":[],"_revision_number":12},{"id":"639cf2931b4d013ad40caffb03dda3554811ba75","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-03-22 10:39:52.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e17c53ed36a52a412e3cc59c3f70176e09183d54","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2022-03-23 14:54:31.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"c9aa15db9d225fc91e409ef0dce69247c6066ccc","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-03-27 09:43:16.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/12/20220327175113/NEC-V-ISCSI : FAILURE in 48m 49s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/12/20220327175113/NEC-V-FC : FAILURE in 52m 00s","accounts_in_message":[],"_revision_number":12},{"id":"193a1b9ee2b3ba2bdadb3a79e51084b7a8f8da31","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-03-28 01:24:48.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"6435b50aaad4a6727a1b435d7d4d80fdbf56b9e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-29 00:32:09.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"e9f529dda30dfca5601c0fa83feddf7acbe132e6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-03-29 00:32:22.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e381ef6f9809aba039fe9e321ddbdd793b4876bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 00:56:11.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc856e06ba3b48d18ecb8567721b2b19 : SUCCESS in 19m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19aab0523be74930b1cb9328a07b2c97 : SUCCESS in 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2fc57238f248d1b316330bb3a34d876e46bde20d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-29 01:03:43.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/13/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f3a1e33/ : FAILURE in 29m 09s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/13/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9529edf/ : FAILURE in 27m 02s","accounts_in_message":[],"_revision_number":13},{"id":"25f12ceb64c26f81b7358fc715de7dd7c93dd70c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-29 01:36:26.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/13/iscsi/ : SUCCESS in 1h 03m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/13/smb/ : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":13},{"id":"2d16174007ad43b8fcb733dd87891807b9d3d00c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-29 02:07:12.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 32m 13s","accounts_in_message":[],"_revision_number":13},{"id":"4d9e3e0a59772ac4bb5dbfb08d50fc2c44f3b66c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-29 02:19:52.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/13/check/iscsi-driver-alletra-6k/850fd19/850fd19.html : FAILURE in 1h 44m 53s","accounts_in_message":[],"_revision_number":13},{"id":"014fe5e3eda70c46cc47896c2fc744aa6f0ef624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 02:51:41.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d0c37d7ce2a4d4dbda61f4e69a4934e : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfbdce2fbafc49658181b06c09101237 : SUCCESS in 12m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d199963600dd4fd9915d482eee1dac4c : SUCCESS in 18m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14ffc89be5bc45449ede01ed54c848fe : RETRY_LIMIT in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6a00389a214f16928e471d46b1fec6 : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/162725c4a4b54bb5a5c7fb2aef9eef19 : SUCCESS in 52m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3fba8d70ad08430da471f17c87405ea8 : SUCCESS in 1h 14m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5522dd89e41e403f957b3962289c5e34 : SUCCESS in 1h 05m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47160f614ba14c42bd34650f31d08b6b : SUCCESS in 56m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce0833c2b3624986ad9eee05bca5cc04 : SUCCESS in 9m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d096a9d767bc47e494fc12759b450b3f : SUCCESS in 17m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f605d229915947fa91cbfa0980c8dfed : SUCCESS in 6m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86d710de847f4221bfbf91ccf86f1d9a : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e3a3f0e94484c48a931af5be3e38135 : SUCCESS in 7m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6b098c760f04b048b4bcc52d7ff0467 : SUCCESS in 9m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4c24f115d7934c93951604377e47b993 : SUCCESS in 45m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c8d63f978e4472188c1be7e0ff163c5 : FAILURE in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/624f75201eae4f22b360c2cc7cfaefec : SUCCESS in 30m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/488f93e5188a4578b35fd8f25d5cb0a2 : SUCCESS in 52m 37s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/3d9bcee10be84fe5b36244866dc1cdbc : SUCCESS in 40m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e2f84b7021b749c5afbfd22bf8d43d82 : FAILURE in 46m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ca33abb977c4794bcef70c637ae4765 : SUCCESS in 1h 04m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e02ba441f7c94f9abe8ca09753c9a88e : FAILURE in 41m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0af51db383f743f0982b6d3a8c709a45 : FAILURE in 29m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/14d2b8ddc23047c797817a969ec98208 : SUCCESS in 1h 37m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f068010987e64792ab92f4f151c207fc : SUCCESS in 2h 14m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6f079a727384bb8be009ffdd35d9a2a : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":13},{"id":"f11acd04f352c5f9c9d0b23cf4917b86f453236a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-29 02:59:24.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/dd11ab950e5c4612831521bc2daadbca : POST_FAILURE in 1h 50m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/50f2d5074ca64a1f8db213ad2b5f6fd9 : POST_FAILURE in 1h 31m 27s","accounts_in_message":[],"_revision_number":13},{"id":"c8fb7e2d93b20c84e8b4a332f8844ac98ad8f3c6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-29 03:32:02.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/750782/13/check/dsvm-tempest-cisco-zonemanager-job/91fa7e7/ : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":13},{"id":"62eb1c71436ddb317f962e876735a52109f9b74e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-29 03:59:15.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-13/ns5_nfs : SUCCESS in 3h 24m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-13/ns5_iscsi : SUCCESS in 3h 26m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"e2081d9c18ab43a9cc9c8f57d0cb3f244af0e980","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-29 06:06:32.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/13/ : SUCCESS in 2h 00m 50s","accounts_in_message":[],"_revision_number":13},{"id":"a99f86e8b5b72325044a60dd7d2d6124cd146b8f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-29 07:06:08.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/13/check/iscsi-driver-nimble/b67b4ba/b67b4ba.html : FAILURE in 1h 36m 27s","accounts_in_message":[],"_revision_number":13},{"id":"568ec0602619cda6e88a4d22e3d8f11d8c229eaa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-29 09:17:42.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/13/check/iscsi-driver-nimble/085c3f5/085c3f5.html : FAILURE in 2h 09m 43s","accounts_in_message":[],"_revision_number":13},{"id":"a750f0421355cb9f31061002439bc7edf23b4278","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-03-29 11:53:54.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/13/20220329173458/NEC-ISCSI : SUCCESS in 3h 17m 06s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/13/20220329173458/NEC-FC : SUCCESS in 3h 18m 53s","accounts_in_message":[],"_revision_number":13},{"id":"a667196e954db58983b81cb50394d4254907c9f9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-03-29 14:50:08.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/750782/13/check/DellEMC_XtremIO_FC/ac83279/DellEMC_XtremIO_FC/861 : FAILURE in 40m 25s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/13/check/DellEMC_XtremIO_iSCSI/263f00a/DellEMC_XtremIO_iSCSI/866 : FAILURE in 34m 49s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":13},{"id":"4960dea07ee32eaa97b459e0c7543e32fb470b5c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-03-29 15:32:22.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"c94a4da05cb4cee84d6c5eebad263fe51ba36e1c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-29 15:38:37.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"a058bef217f155e355302ce003d199738ce5f101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-29 19:23:42.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"ec87670c8dfef534d4b6c0e90a0c4627ac5ef339","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-03-29 19:23:57.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9c4b8fbb008c0303446095513fe14eabdb93ef4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 19:24:29.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"bf93380c02812e1f583d2a5374c6903f9697982e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-29 20:38:05.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/14/iscsi/ : FAILURE in 1h 13m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/14/smb/ : SUCCESS in 1h 12m 36s","accounts_in_message":[],"_revision_number":14},{"id":"3a033c6a87b2c536062881052f0c6279d51c1b35","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-29 20:47:16.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":14},{"id":"870dddb40a5bb1719eb74c881e4390fb2a74fd2f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-29 21:02:18.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/14/check/iscsi-driver-alletra-6k/48062b9/48062b9.html : FAILURE in 1h 36m 24s","accounts_in_message":[],"_revision_number":14},{"id":"c62235909570add0dc22ec3e019a4f439e360330","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-29 21:41:00.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"fa765db3eea8b174f20721e7ffd1bfc69c880b31","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-03-29 21:41:00.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":14},{"id":"467bfea7671f028d56b7bc0f92235b687c9eab38","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-29 21:41:02.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"49a8a94a7b62ec11e13b3561a8a6fc848245951d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-03-29 21:41:03.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"378ec289738bb8ef175f90289cc64041c3c23f29","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-29 21:41:04.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"464202ab4a76c081b979b0a174a4478e6c6c16d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 21:41:04.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"160f09c9c40247bbb20955f9f327ec89d0beca67","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-29 21:41:06.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"8583ae8aebee42d4bc55ed3da42dc9e56c6c85cf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-29 21:41:06.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"dd5b5ea3cce210edd3f887c70d375816506b185b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-29 21:41:08.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"cb078eb413ba595a740a8bf3bdca93f15dd1d3b6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-29 21:41:21.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"a29398c07efa41c7c2b98ebb59a310bb6820c824","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-29 22:49:14.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-14/ns5_nfs : SUCCESS in 3h 25m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-14/ns5_iscsi : SUCCESS in 3h 23m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"f56b74b9030ea34897a0637d8181349025a58a0d","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-30 00:40:41.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7afeb11505d13c7d36168962aa8cb14edb99fafb","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-03-30 00:40:53.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d97f479b6dfc74071d5b2ee09f741f0b3c64dede","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-30 00:41:22.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"9c11f09afba9de4317895aa9e3b6176d1dc1b59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 00:41:22.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder-tempest-plugin for 835512,4","accounts_in_message":[],"_revision_number":14},{"id":"a8c116f437e43d866e9eb0bbdb5513f5afb2cfba","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-03-30 00:41:24.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"6a422155ec31b57442a00f5e942acdc2c60b3402","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 00:41:26.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder-tempest-plugin for 835512,4","accounts_in_message":[],"_revision_number":14},{"id":"1ec3dc5a6953098ad4c62acc75d93759532614f9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-03-30 00:41:26.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"bba957843e40422f6f68040262ec5ef492bea7ad","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-30 00:44:36.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"aab14500d4119edca6ad3ad01797582be81cfc12","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-30 01:41:35.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/14/iscsi/ : SUCCESS in 59m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/14/smb/ : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":14},{"id":"77ca6a2b543a86ed89b58c750173392fba17a876","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-03-30 12:20:47.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/14/20220330190156/NEC-V-FC : SUCCESS in 2h 11m 18s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/14/20220330190156/NEC-V-ISCSI : SUCCESS in 2h 18m 47s","accounts_in_message":[],"_revision_number":14},{"id":"0a6cc2fcd98c5e47fbc0153dd15b899da670366c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-03-31 02:46:57.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/14/20220331082919/NEC-ISCSI : SUCCESS in 3h 17m 22s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/14/20220331082919/NEC-FC : SUCCESS in 3h 17m 35s","accounts_in_message":[],"_revision_number":14},{"id":"a864dcdbb807c9efad64c13d550f390ef9c0387b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-31 13:09:14.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"32678ce3fe7b2fa7e793d7782b7432d47a213317","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-31 13:09:26.000000000","message":"Patch Set 15:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15673/","accounts_in_message":[],"_revision_number":15},{"id":"ca81e8c208e52ac5f060fb764bf1396f4ba48830","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-03-31 13:09:29.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"7f006c3fce0d65d22dd82a32284019cb908f2008","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-31 13:09:37.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"98bbcb4528fb7cffd9c6ded4b026665f252f60b8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-03-31 13:09:51.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0c430958d64ae6b0e6a773603548c1b684eef20c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-31 13:11:09.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15673/ : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"1fffb4a67f550ce6bf7d2fa9718a03ccb09b225f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 13:50:45.000000000","message":"Patch Set 15: Verified-1\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e4846f14e2440e7b294f88b0a583372 : SUCCESS in 24m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5f0b94d86a34a2ea8e96ca71f6a71c5 : SUCCESS in 26m 07s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d97df5e32633b99d2aed1c766cd5d648b8c16543","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-31 14:15:25.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/750782/750782/15/check-not-vote/ext-cinder-zuul/26de350 : FAILURE in 1h 02m 43s","accounts_in_message":[],"_revision_number":15},{"id":"ac24c0bed992feddd3b401b7608ba2782ea8ee9e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-31 14:35:52.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/750782/15/check/dsvm-tempest-cisco-zonemanager-job/d9d0f08/ : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":15},{"id":"4a0281daee4478cf7661bc6e3ac88367aed61a7b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-31 14:50:27.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : FAILURE in 1h 00m 15s","accounts_in_message":[],"_revision_number":15},{"id":"3c82d780fe22985512cbbcb4d573dac507b484bd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-31 15:09:53.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":15},{"id":"7b1ecaca84e65bd7ae9dfa094d2434fb0f629fa6","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-03-31 15:14:01.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/06dd72a/index.html : TIMED_OUT in 2h 02m 57s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/505a8be/index.html : TIMED_OUT in 2h 03m 09s","accounts_in_message":[],"_revision_number":15},{"id":"7e798a4b4b8e9d06b83466f287ff6b46bc6b1c39","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-31 15:18:01.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/15/check/iscsi-driver-alletra-6k/75afc07/75afc07.html : FAILURE in 1h 38m 36s","accounts_in_message":[],"_revision_number":15},{"id":"73a9cce24116152c4df48e4b53089e297f1f2f54","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-31 15:49:13.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/15/iscsi/ : FAILURE in 2h 30m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/15/smb/ : SUCCESS in 2h 38m 30s","accounts_in_message":[],"_revision_number":15},{"id":"5b69fea29dda8da16b9eebdb7c3c24e09f4a837c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-31 15:53:06.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/15/check/iscsi-driver-nimble/ff91286/ff91286.html : FAILURE in 1h 01m 46s","accounts_in_message":[],"_revision_number":15},{"id":"6045c27d479d69cc2c93353b226fd36b388eab78","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-31 16:00:53.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5fdd549d504743679740ee0b1d47967a : POST_FAILURE in 2h 10m 02s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ab742719d6d24fc2b05631438d3f0a5d : POST_FAILURE in 2h 47m 22s","accounts_in_message":[],"_revision_number":15},{"id":"b339fe9984b12a80f594f292c8d8e914c7f8fa36","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-31 16:01:04.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/edfeadc/ : SUCCESS in 1h 56m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0f5d8b0/ : FAILURE in 2h 44m 18s","accounts_in_message":[],"_revision_number":15},{"id":"3ed90601051a7a8b32cf39a13a0f1afa0cb3623c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-31 16:37:35.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-15/ns5_nfs : SUCCESS in 3h 25m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-15/ns5_iscsi : SUCCESS in 3h 27m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"8460a426b72fef28d444ee82a60b56882e46e5ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 16:42:12.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380cfbaafa4343219b7f93528c239937 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c7dc4e0fc49473998e1962814116454 : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/182b1cc19ef54fdea5de0b7882619131 : SUCCESS in 14m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7938dd7495574c31a8e65111557e4900 : RETRY_LIMIT in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f42f8b1c5bc4cb1b5dd8f4a93575c7e : SUCCESS in 11m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/ba1cdef395c84b9d8db9e410b8f37d5f : SUCCESS in 1h 06m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acca4a4c2cde4aea8728724912eda828 : SUCCESS in 1h 16m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfedd2a3cd594e1fa03a21c693d1c800 : SUCCESS in 58m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ff7a50c29bd4cb7b88ea83c37b81840 : SUCCESS in 9m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29b33e452b2f41c0a1c6cfbc6fff43a8 : SUCCESS in 16m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/04ddd7bec5684f9cb0d68f40ecd9af51 : SUCCESS in 5m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/56c12d9f5d844c538230a5ef3decf2e7 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9997eb8079234b2191bbf2fa95837202 : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58da4068e43f430f8caaa7623c22fee6 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bb51eb10f4a347f7abe6b9eb93e7527c : SUCCESS in 42m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1dc5250b189e4a6c81d1659ca4f162ec : FAILURE in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e46f7713ff70419a8f18fc21ed096cad : SUCCESS in 1h 26m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9931720049944e8caf18206e61c5cfe9 : SUCCESS in 1h 33m 45s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/bc2ae45f83a547c893575c5535f8aefb : SUCCESS in 2h 06m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/535eb3408faf462c86348f14d582f7e2 : SUCCESS in 1h 58m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ff005161bbbb4e00aa7985c490995116 : SUCCESS in 1h 07m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/44bfa048230f4cdabb5b54649b91c674 : SUCCESS in 1h 37m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9520da9955e4461f9af7a8e28c9d3f02 : SUCCESS in 45m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c6cbedf480a427e83060d2b7df55b07 : SUCCESS in 1h 17m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e8bfc5c821204b69b8caa9b5e60f51f2 : SUCCESS in 2h 56m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cab7867795f442069ba8921d5038e75a : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":15},{"id":"5d189bd5170a7dbf1b46080fa38ebe9b8feac0c6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-03-31 16:43:48.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/15/upstream-solidfire-check/cinder-solidfire-iSCSI/f0a52a6/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"34ce1d257873f569a22075c7bd210d4ed2ef2cc3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-31 16:54:22.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/15/check/iscsi-driver-nimble/f999f0c/f999f0c.html : FAILURE in 59m 29s","accounts_in_message":[],"_revision_number":15},{"id":"14eaf4c4142345f5690f59422ca524aa34d2125c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-31 17:06:28.000000000","message":"Patch Set 15:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/31_03_2022_16_42_19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":15},{"id":"0c9b4f4455ee589b93ae47d9eeabcfdf778fa1a3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-03-31 18:53:05.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/750782-15/HPMSA-iSCSI/ : SUCCESS in 52:27.77\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/750782-15/HPMSA-FC/ : SUCCESS in 46:02.43","accounts_in_message":[],"_revision_number":15},{"id":"ec3d4aba9c43ed950e47e99e0347a7706f391c35","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-03-31 18:56:28.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/750782/15/check/DellEMC_SC_FC/d7f51b5/DellEMC_SC_FC/380 : FAILURE in 1h 13m 48s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/15/check/DellEMC_SC_iSCSI/46a0961/DellEMC_SC_iSCSI/402 : FAILURE in 2h 13m 54s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":15},{"id":"b082fb4327cf5e8d48e01f5b08f59203f9638a16","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-03-31 19:18:44.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/750782-15/Lenovo-iSCSI/ : SUCCESS in 1:05:51\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/750782-15/Lenovo-FC/ : SUCCESS in 59:24.25","accounts_in_message":[],"_revision_number":15},{"id":"c34b4aed0f123c56dda086ef200f8e400ddd4ff6","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-03-31 19:21:32.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/750782-15/PVME-iSCSI/ : SUCCESS in 1:04:30\n- PVME-FC https://os-logs.seagate.com/dell-ci/750782-15/PVME-FC/ : SUCCESS in 1:03:36","accounts_in_message":[],"_revision_number":15},{"id":"c99f35ce9e90daff4a8c2d4b7390889ed1cc92ad","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-03-31 19:29:56.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/750782-15/Seagate-iSCSI/ : SUCCESS in 1:09:30\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/750782-15/Seagate-FC/ : SUCCESS in 1:07:14","accounts_in_message":[],"_revision_number":15},{"id":"7419f9bfc4c1687b5961475153d301e556ea3855","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-03-31 20:20:26.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/15/20220401030616/NEC-V-FC : SUCCESS in 1h 46m 20s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/15/20220401030616/NEC-V-ISCSI : SUCCESS in 2h 14m 07s","accounts_in_message":[],"_revision_number":15},{"id":"cb23c1599a69672de730c1f73bf729946579b46b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-31 23:42:42.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/15/ : FAILURE in 1h 40m 17s","accounts_in_message":[],"_revision_number":15},{"id":"c257945b01a934c1ded19ca5c53fc23402ddd26d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-01 11:32:24.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/15/20220401171013/NEC-FC : SUCCESS in 3h 17m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/15/20220401171013/NEC-ISCSI : SUCCESS in 3h 22m 08s","accounts_in_message":[],"_revision_number":15},{"id":"816ad87c8ba12ee5175f0a3c4a3c59cc993b17a8","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-04 17:20:52.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"50ed86d2c4820960409aa35413386d26653b81df","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-04-11 06:57:34.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/750782/15/check/DellEMC_XtremIO_FC/035a01b/DellEMC_XtremIO_FC/881 : SUCCESS in 1h 17m 08s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/15/check/DellEMC_XtremIO_iSCSI/6034bf8/DellEMC_XtremIO_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"69a51a42ba95719a43c10c89b4ec765ac1482475","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-12 13:39:36.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":15},{"id":"17ad98328771ab55575baa5599f51a6ab5be9f23","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-22 14:41:39.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3d1a18cb520e80d045ed385ec34740059e000ef0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-22 14:43:24.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":16},{"id":"90dcaaceaed8f83f6cf712c9e0c04b85d3404856","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 16:11:51.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cee9327b749744a6b8bd2145568da1f4 : FAILURE in 17m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87e111e595424ecfa99ba8a371a26025 : FAILURE in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e414ee01729af8da96897694b44c179892c47a61","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-22 16:42:27.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/16/iscsi/ : FAILURE in 46m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/16/smb/ : FAILURE in 46m 17s","accounts_in_message":[],"_revision_number":16},{"id":"27f42568394f93271e5137f9bc855d1befef11dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 16:42:42.000000000","message":"Patch Set 16: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":16},{"id":"29ebbac648a3312208f9bdce4719c279f29d0bad","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-22 17:26:01.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"e5b10a90a6f483da6477e0ac1990d61db2225c43","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-22 18:06:10.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/16/check/tempest-dsvm-full-focal-py3/2976fac/ : FAILURE in 55m 58s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":16},{"id":"b7b07fa45beaf4a0d509254702b4642e381dfc1c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-22 18:20:01.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/16/check/iscsi-driver-alletra-6k/78a4692/78a4692.html : FAILURE in 1h 00m 47s","accounts_in_message":[],"_revision_number":16},{"id":"351fc0732d5cf5145f152b6d57c1ddc3dfb107b5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-22 20:51:06.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-16/ns5_nfs : SUCCESS in 3h 30m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-16/ns5_iscsi : SUCCESS in 3h 25m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":16},{"id":"4d5c0ad5cc8594328632c3bb81a87c07a4d39432","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-22 21:31:33.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : FAILURE in 1h 20m 47s","accounts_in_message":[],"_revision_number":16},{"id":"34c8c993fe28ee1d3b251bae86baa7fe3e6e3446","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-22 23:17:09.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/16/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1379cd5/ : TIMED_OUT in 3h 02m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/16/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8929fcd/ : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":16},{"id":"a0f9ec9d5efbbc1ed9d711da18cdd902c1701326","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-23 05:34:26.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/16/check/DellEMC_Unity_FC/c4bef0d/DellEMC_Unity_FC/665 : FAILURE in 1h 17m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/16/check/DellEMC_Unity_iSCSI/59cdec9/DellEMC_Unity_iSCSI/562 : SUCCESS in 1h 46m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"9aecf93aa65879d0e3b638d9b3ef3f05035a0457","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-24 11:10:00.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/16/20220424172952/NEC-V-FC : SUCCESS in 1h 58m 07s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/16/20220424172952/NEC-V-ISCSI : SUCCESS in 2h 40m 05s","accounts_in_message":[],"_revision_number":16},{"id":"c68bef4600dae9501615c6e1cdfe9b1f3b47de68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-25 13:22:56.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"d80951fdcbe82669791127ada33ea0e281b28ad4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-25 13:23:07.000000000","message":"Patch Set 17:\n\nBuild Started","accounts_in_message":[],"_revision_number":17},{"id":"e1911842aeb3d9817a8595868a076898f1923706","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-04-25 13:23:12.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"2569c85845443d42b4e58386e171198f487db96a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-25 13:24:19.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":17},{"id":"0faef1e9951cb2c0c4be207aa0126e7ba4d489ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-25 13:56:47.000000000","message":"Patch Set 17:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1650892986271 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"7ef7df4062139f21c802a03c48baf7412ec99690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 13:59:08.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e14381275de74bb6a90b38bcb00ec406 : FAILURE in 32m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71214492a3c74322b16f7fa4d9785c78 : FAILURE in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e3fc99e5dec0d2950db937787c4b8d463551a0da","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-25 14:21:16.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/17/iscsi/ : FAILURE in 57m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/17/smb/ : FAILURE in 47m 36s","accounts_in_message":[],"_revision_number":17},{"id":"5496d48ea97aee1c692b706b298e3d38edcbd43b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-25 15:00:57.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/17/check/tempest-dsvm-full-focal-py3/7030be8/ : FAILURE in 1h 34m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/17/check/devstack-plugin-ceph-dsvm-focal-py3/9b9ab27/ : FAILURE in 1h 34m 53s","accounts_in_message":[],"_revision_number":17},{"id":"7e66b7054d4bcd63f14dd12e237f6a39b65f21d4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-25 15:08:02.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/17/check/iscsi-driver-alletra-6k/1d39d2a/1d39d2a.html : FAILURE in 1h 33m 45s","accounts_in_message":[],"_revision_number":17},{"id":"eee1af696e6185e9a445eb6423c778e1d6d0a81c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-25 15:44:01.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":17},{"id":"3816cbff0a3f8d25a0bd70e82910ec17b36d83f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:55:10.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89fb369996054edba8e4a234707460f9 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca172630cb234b6fa96e1f0a56e21fc8 : FAILURE in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd0c33b5e3014b85b3edb2655ed61ad9 : FAILURE in 12m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b212e40ab914b7f87893975d38efa77 : RETRY_LIMIT in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297eabf676694a7a8b747e7648526ecd : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/7219b817ea6240268cab3bdcdefc2d19 : SUCCESS in 1h 02m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/531344a2bff34280af919fc111da3405 : SUCCESS in 1h 21m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3ef82a35d944435829cb039045232b9 : SUCCESS in 52m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60b13753cfee4f5e8862e90fb47f66f6 : SUCCESS in 8m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0a22d4f7eb8443f2a86448b8c8dcbd23 : FAILURE in 12m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1d7e095d92b54d5db986342eb84a4024 : SUCCESS in 6m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2e063da2d2784e8a9257b3929c075ed6 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9372436216314134a9509ddb041ba65b : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d9e128d9d7b4d25bd90a46047c93dbb : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3825d0e85c264b3f946c35e582df8f40 : SUCCESS in 43m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/250969ed602445e389f2a9b32eb1fccb : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ac3248901d504d748e3ad3db1f348e98 : SUCCESS in 1h 13m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ba6c4cdcf294dc0a5b94dffca9bfe6f : SUCCESS in 1h 26m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dbcf8fc66ee84f018fffc1eda96212c5 : SUCCESS in 1h 42m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e03a5bb72636459fb6d12a76447e1323 : SUCCESS in 1h 08m 25s\n- cinder-tempest-lvm-rbd-multibackend https://zuul.opendev.org/t/openstack/build/b46c7f68e40544389fa4888f6385d4e5 : FAILURE in 19m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fbfcaa4f76ff432e94a7e6c26d37ea66 : SUCCESS in 45m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3f38ec6d21a7478a82d4c503f89370e8 : SUCCESS in 1h 14m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf88cfa9eef947ce9e2b0ce746f93443 : SUCCESS in 2h 26m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17b57d0d42ff457c8e684ac66ada6aed : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":17},{"id":"16c969ccc238bee7a95e61eb245466a7b1893377","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-25 16:28:49.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/17/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3c24907/ : TIMED_OUT in 3h 03m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/17/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5138da4/ : SUCCESS in 1h 33m 32s","accounts_in_message":[],"_revision_number":17},{"id":"6e613ea7e1bfac7bdaf0f9f3fa49c94f2f4f8c1e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-25 17:31:16.000000000","message":"Patch Set 17:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-17/ns5_nfs : SUCCESS in 3h 24m 23s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-17/ns5_iscsi : SUCCESS in 3h 26m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":17},{"id":"3adb1c30c593ffb19e68a2e9eb421b02b8568bdc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-25 18:01:54.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"4493e96e10b0260c9064bfc30f4b8740939355fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-25 18:02:02.000000000","message":"Patch Set 18:\n\nBuild Started","accounts_in_message":[],"_revision_number":18},{"id":"e5b163dc3d6091f59006aa8c89200babbec0ae96","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-04-25 18:02:09.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"54d1aa77d03241534429775d2cf9cd49c4030895","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-25 18:03:15.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":18},{"id":"ebaa7b2d7ad210bee025ae6d0f7abfb605bc79bd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-25 18:07:09.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/b970569ac64946d59d0398e609f9ca36 : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":18},{"id":"87e1eb852ad955cb06c0e71363b6d4094b50e311","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-25 18:18:49.000000000","message":"Patch Set 18:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1650909721751 : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"a1a9757f950142242f05444abe8e17d88a9c5620","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 18:28:04.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a9eb950e7c84f50b89aafe1c351ec18 : SUCCESS in 20m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be783457f331430f86f135b25cd8370b : SUCCESS in 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"02c5f49f021110367d96ca859f05654b4fbd0ad1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-25 18:51:02.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"b13acdaab4d968249349ca8d16f975a5418507e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-25 18:51:12.000000000","message":"Patch Set 19:\n\nBuild Started","accounts_in_message":[],"_revision_number":19},{"id":"2857cd032c280906ff11da593a99bc8de41a972d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-04-25 18:51:18.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b25f611a29c2f32fda9962548818366ddbe276a7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-25 18:52:26.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":19},{"id":"41fb64a39289b6c995a72f37d060df934e821c7c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-25 18:56:14.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/e8c8bd38cb1946a6964850c805f19e42 : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":19},{"id":"e5a2db646825110857569b90b9efcb6c0af7bacc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 19:14:22.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9f5dbeae6b4491d9cb1e50f60c7f956 : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82ec98e63c5747fe88baa79f06ddf668 : SUCCESS in 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"38e377d25532edb453ca5760426277c6e72944c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-25 19:23:24.000000000","message":"Patch Set 19:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1650912671826 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"b83a42ee7c1c65f455194a3fcbd8a30a23dde1ba","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-25 19:48:00.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/19/iscsi/ : FAILURE in 46m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/19/smb/ : FAILURE in 56m 18s","accounts_in_message":[],"_revision_number":19},{"id":"c7c58a3eb6649bfb3952cbd18b8fda22de60e789","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-25 19:54:50.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/17/20220426023957/NEC-V-FC : SUCCESS in 1h 48m 13s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/17/20220426023957/NEC-V-ISCSI : SUCCESS in 2h 14m 50s","accounts_in_message":[],"_revision_number":17},{"id":"f2a1d568715beac989071e8a1ff5d7ac0e6e84d1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-25 20:26:15.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : FAILURE in 1h 29m 59s","accounts_in_message":[],"_revision_number":19},{"id":"7443df10e656bd42b9e349c012b5da788de86ae8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-25 21:09:38.000000000","message":"Patch Set 19:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/19/check/tempest-dsvm-full-focal-py3/7805607/ : FAILURE in 2h 12m 26s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/af4f9d463a5149df90ca43c0a8fb1e69 : DISK_FULL in 2h 14m 12s","accounts_in_message":[],"_revision_number":19},{"id":"fc582b35ead89eaebab9ee194b7c7a57904602a0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-25 21:31:37.000000000","message":"Patch Set 18:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-18/ns5_nfs : SUCCESS in 3h 26m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-18/ns5_iscsi : SUCCESS in 3h 29m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":18},{"id":"deee0ea9deed8bf47f6a45edb5f2343afe0092b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 21:34:51.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d5c5633baa645c695809498670701aa : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58e9f1b3f83241e9afb9f7ad430dbee3 : SUCCESS in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a776f2ef5ce6481aa0f20415eb346fad : SUCCESS in 12m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/939d09d310fe4711884f2956c0f0df28 : RETRY_LIMIT in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e37c3133674570a2e871a05f20a22c : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/b5aa565e6d8f40db80950162297fe43a : SUCCESS in 51m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69c2314cd7ba412882899fa284c7085d : SUCCESS in 1h 42m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af183465ca334907b6b8390f5285098b : SUCCESS in 44m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d8229b3d0f4620a861b537387ca84d : SUCCESS in 8m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/09af1cae06344dfbaa95924a98dd5b63 : SUCCESS in 15m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b4ee578dc2034a8583ecfb82cf988dfd : SUCCESS in 5m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9881d16c523943e7bbe252cfab073739 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3db4030fc114747bfb564d105a245bb : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/84edde7757894689ba588ec7132d51f6 : SUCCESS in 8m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dcf3c58eac51495c9f72c342f1d379ff : SUCCESS in 34m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc71da01c7794caa8d89b3ac082859a0 : FAILURE in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b339a135994f40a0a6a090be366daeb5 : SUCCESS in 1h 14m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6ce0f40f70d047b9bd4d54274ad36604 : SUCCESS in 1h 39m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b4a7b79771304c01b869fbd3b335603a : SUCCESS in 1h 43m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f332f4e3e8e54fd49eb3d84d79a230a6 : SUCCESS in 54m 46s\n- cinder-tempest-lvm-rbd-multibackend https://zuul.opendev.org/t/openstack/build/f909a30ce93b42aa88d1e929294a37cc : FAILURE in 57m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1c64413ce2cd4e46892f3a867d0f3f4d : SUCCESS in 31m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7b4cb26c54574b0595fc15ac462551ff : SUCCESS in 1h 29m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c1539b640204d37b63349ca8d6db78f : SUCCESS in 2h 37m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19baf73c6a37446481d54a8315188138 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":19},{"id":"497f501e63a13650279dd20f536633f6cebe62a0","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-04-25 21:51:51.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-82-750782-19/1650922491 : SUCCESS in 16m 58s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":19},{"id":"ab30013437e6f7dcae70ae2142c39f5c812d47d4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-25 21:55:53.000000000","message":"Patch Set 19:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/25_04_2022_21_35_02\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":19},{"id":"6085c2f44fd3ca80d32244afad67ec6bb815475c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-25 21:56:21.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/19/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ae6d7a8/ : TIMED_OUT in 3h 02m 59s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/19/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1455b03/ : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":19},{"id":"7eaa8e7ca09930e1010cbc978797dd6b133b8f60","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-25 22:42:08.000000000","message":"Patch Set 19:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-19/ns5_nfs : SUCCESS in 3h 27m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-19/ns5_iscsi : SUCCESS in 3h 29m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":19},{"id":"bb046e3a40516307c34877cb728ac4dcfdf98c34","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-25 23:40:33.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/750782/19/check/DellEMC_PowerFlex/c3e32fd/DellEMC_PowerFlex/652 : SUCCESS in 1h 17m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":19},{"id":"bdcb800449254820a612f772d987946337161914","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-04-26 00:07:22.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/19/check/DellEMC_PowerStore_iSCSI/fa3532e/DellEMC_PowerStore_iSCSI/448 : SUCCESS in 1h 22m 47s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/750782/19/check/DellEMC_PowerStore_FC/c5b0999/DellEMC_PowerStore_FC/463 : SUCCESS in 1h 10m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":19},{"id":"5cff1f6e173e9d26ae287217601172947929c73c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-26 00:51:18.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/19/check/DellEMC_Unity_FC/9fc522b/DellEMC_Unity_FC/678 : ABORTED in 3h 00m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/19/check/DellEMC_Unity_iSCSI/b83dc6a/DellEMC_Unity_iSCSI/573 : SUCCESS in 1h 38m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"9fc8fc1751c2cdc5a44a1e05ea75a30b35146f42","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-26 01:19:33.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/750782/19/check/DellEMC_SC_FC/523f2b3/DellEMC_SC_FC/525 : SUCCESS in 59m 28s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/19/check/DellEMC_SC_iSCSI/c65f262/DellEMC_SC_iSCSI/546 : SUCCESS in 1h 50m 11s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":19},{"id":"979a423af17fe60106fa54ab6b0588eccf7d31ef","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-26 02:07:28.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"79cd5a0ada2e9618f98499544f6169094b86d39b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-26 02:15:38.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : MERGER_FAILURE in 23m 23s","accounts_in_message":[],"_revision_number":19},{"id":"46558ec77ce4ce2a826a7bba385b76eedf9050ed","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-26 02:22:58.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/750782-19/HPMSA-iSCSI/ : SUCCESS in 51:30.92\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/750782-19/HPMSA-FC/ : SUCCESS in 45:38.39","accounts_in_message":[],"_revision_number":19},{"id":"34c8e33bece6efe6e8aa915c0702fb1e75ecde83","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-26 04:25:36.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/750782-19/PVME-iSCSI/ : SUCCESS in 1:03:42\n- PVME-FC https://os-logs.seagate.com/dell-ci/750782-19/PVME-FC/ : SUCCESS in 1:01:32","accounts_in_message":[],"_revision_number":19},{"id":"f1de4db5a97efde38349e0cfdb7485cfad76735a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-26 09:34:28.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/19/20220426122232/NEC-V-FC : SUCCESS in 1h 43m 21s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/19/20220426122232/NEC-V-ISCSI : SUCCESS in 6h 11m 53s","accounts_in_message":[],"_revision_number":19},{"id":"576e74f976c0fc07b3ed861117cd39e56abad5a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-28 19:50:54.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"afdcb54911a311416a7afc155a8e345f318f36cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-28 19:51:05.000000000","message":"Patch Set 20:\n\nBuild Started","accounts_in_message":[],"_revision_number":20},{"id":"622b960be2711bf944a33eefdb2379ceb60a5c91","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-04-28 19:51:09.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"c15e6042e132cfb149289fac4515e7649b974ee8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-28 19:57:25.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"5f013845c36b20002a6b5f1ffd432ab3fca2fe83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 20:22:16.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80abefa215e5425a955271ee25eeef50 : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e155ea738b4245ae983f765a31e0d3c1 : SUCCESS in 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"5dbfe47a1eb7e18e1d215ea7493d7f72792fd850","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-28 20:25:05.000000000","message":"Patch Set 20:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651175464376 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"6e376178ca9fe696f8e38534a6b553d4ad6311de","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-28 21:10:34.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/20/iscsi/ : SUCCESS in 1h 19m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/20/smb/ : SUCCESS in 1h 16m 00s","accounts_in_message":[],"_revision_number":20},{"id":"a0a5e0ba34026474d00bede7a5852e722297bf22","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-28 21:13:27.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":20},{"id":"cbbc4c78e49e112d29f123838257891a328ef7bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 22:04:48.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2619a69fc9ed42d5986ed48e59b9e8e7 : FAILURE in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a0a970bec47466181578b2808c4ebc2 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d4fc310353d4c73bd8961e7c9ce9e3c : FAILURE in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3c2ddfae60649cd835e8327275bc1e2 : RETRY_LIMIT in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985ed5f1de8643adb5ddd98b3391cbc6 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/d734db59726b44808638439fa730fbfc : SUCCESS in 1h 09m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cccfd7da17c5427b9195e871556b467c : SUCCESS in 1h 04m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/306a112d2b5a4b5b8cb3a6e28aaae1d1 : SUCCESS in 1h 01m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c98dd9820a044dd6bf7fd47b04b1f3b6 : SUCCESS in 8m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0e073251cff9491fb752c0002cc76af9 : FAILURE in 13m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0f6c56b192fd4810bf72d9993da91050 : SUCCESS in 4m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1b20aeb8be014e728d45fd8734b25ee2 : FAILURE in 4m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9b39e3581b446c38bbee45411f721d3 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5280cd19a3aa447ba6e46023d5be0d86 : FAILURE in 4m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e371cedda741411594d6b3a1fed5ebf7 : SUCCESS in 50m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ffb68386c5940d097830ef669b4b02b : FAILURE in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5e220e06275495ea775aeeb9b61cb71 : SUCCESS in 1h 12m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e4dfc036177745489fe930b247a77a05 : SUCCESS in 1h 54m 28s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/9121b48357a540c1bd156ff6864c721b : TIMED_OUT in 2h 06m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6f027c49a684b02b596c88255a7acb0 : SUCCESS in 1h 24m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad65581adc434a948cb272ec95572df2 : SUCCESS in 1h 21m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/261aebbbb97f4c599401166f7376484e : SUCCESS in 1h 19m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/33c755171a2545a59623e0d9a49e5152 : SUCCESS in 35m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1e272b578495414a838dbcf78673d340 : SUCCESS in 1h 13m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8296ee2ec141420ea697aba35e8f65bf : SUCCESS in 2h 09m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13995ad6cf864ba9b961ff8e5729ec1a : FAILURE in 1h 00m 31s","accounts_in_message":[],"_revision_number":20},{"id":"56807c6285fb781c18394d2b2e323a6bba1dcb2c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-28 22:15:35.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/20/check/iscsi-driver-alletra-6k/74f1fa1/74f1fa1.html : FAILURE in 1h 35m 04s","accounts_in_message":[],"_revision_number":20},{"id":"771a4aff8ee542f1f59fc7a1c5c5bee94ed6f856","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-28 22:24:13.000000000","message":"Patch Set 20:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/20/check/tempest-dsvm-full-focal-py3/a2437d1/ : SUCCESS in 1h 59m 02s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4e5e90a419c744d9af6229494ad1e9de : DISK_FULL in 2h 18m 27s","accounts_in_message":[],"_revision_number":20},{"id":"fd7e37b1ba1ed8fe27a991781ff81464936c66af","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-28 23:08:08.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/20/check/iscsi-driver-nimble/023e08b/023e08b.html : FAILURE in 1h 00m 36s","accounts_in_message":[],"_revision_number":20},{"id":"e8e0ab812caff4bdc7601b758f101d98dfbb9a46","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-28 23:39:28.000000000","message":"Patch Set 20:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-20/ns5_nfs : SUCCESS in 3h 26m 50s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-20/ns5_iscsi : SUCCESS in 3h 27m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":20},{"id":"bcab0cb0b818790f462af19ebc7a472fb531a49f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-29 00:31:06.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/20/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d5973be/ : TIMED_OUT in 3h 03m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/20/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0dbc300/ : SUCCESS in 1h 31m 07s","accounts_in_message":[],"_revision_number":20},{"id":"98ed618c3fed58c4d650bececbed99aab4d7eee7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-29 13:11:08.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"d2ed9f2985bda61c97c6a832190cbf04c6408508","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-29 13:11:16.000000000","message":"Patch Set 21:\n\nBuild Started","accounts_in_message":[],"_revision_number":21},{"id":"81b2ef364104f5650e6cff5a6b2b6969e7943666","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-04-29 13:11:23.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"b02472495b791ba900b771ac60526a0c65002c4a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-29 13:19:56.000000000","message":"Patch Set 21:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":21},{"id":"a8b4a61529ad71701317d2042253ddb804c80e75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 13:33:59.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/244964f579844e9cae298af6bf1e2936 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41fc46e2a3da42b7a0c2db530fb827eb : SUCCESS in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"faeeb737df406773a23e96d8b4f81d55e20778a1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-29 13:44:23.000000000","message":"Patch Set 21:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651237876132 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"7df35addba8d2057465b2c0aa87b9acd06fb1c0b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-29 14:14:45.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/21/iscsi/ : SUCCESS in 1h 02m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/21/smb/ : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":21},{"id":"86c30ac51232c79e44231d1e42da1808afda11c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-29 15:31:42.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":21},{"id":"7ece550038f49a67011cd876ebb9538f4117b54c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 15:56:59.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb34637de5444e9864a4b82a634b226 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59f5ea27962a4d1e9516dffea04b6369 : SUCCESS in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/400e86e8d04b4b929f9c73254abdc51a : SUCCESS in 11m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a24025626c284c25a7667e9ebb8c0d5a : RETRY_LIMIT in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0fa93ed0b94080ad37e71a80b0b458 : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/9d70d12f399640dd80a8e297f477e271 : SUCCESS in 1h 09m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/694cbeb1bb0649bdb091c11d42fe0513 : SUCCESS in 1h 24m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49da3eb74aea4210ba7087ff6c93da4b : SUCCESS in 55m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03135d06f89f401c8610565b8c57ca67 : SUCCESS in 8m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cdaf13a2f76d48c5b22c3fb7b4e52d36 : SUCCESS in 15m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc650838e0984b109aa68f25186982b8 : SUCCESS in 6m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5cb80a612c18429f9573f7f9fb36ffe2 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c11592c2113141b7990efd6f51b99552 : SUCCESS in 6m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b97e6fdfc51456ca073b8dcb8958aa8 : SUCCESS in 7m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9667fde6b1464be7bfd03bda4ee86fe4 : SUCCESS in 51m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c756bbfb62bc46f481ea064364022ee8 : FAILURE in 5m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3de13b0ebcbb472f823a0ddd2ce298b2 : FAILURE in 21m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/31d33d683f0349078a9e462e2f24841f : SUCCESS in 1h 31m 29s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/b2cc1de5b32a445386c39c563be89bfc : SUCCESS in 1h 38m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/991332af11754ff080b84104caaecb55 : SUCCESS in 1h 33m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55897cee6f9448e191b1c9c9f6b92d0e : SUCCESS in 1h 13m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dab363417313485fba7c9f4c9d9f2c9d : SUCCESS in 1h 07m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dd2ba4ce476c437db32dd3a1aca89fd7 : FAILURE in 53m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8084908c7c014dc18c3734deaf1aa086 : SUCCESS in 1h 12m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1643ea6c3d834b07812f27c1de733a23 : SUCCESS in 2h 38m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a730f6ab4bf427d8f69b2911203ce62 : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":21},{"id":"4d2ed644a49f669911415efd2d766574e73c9771","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-29 16:16:21.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/21/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2d2939c/ : TIMED_OUT in 3h 02m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/21/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/312770a/ : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":21},{"id":"544bd6a8a446a669d41db227135186e21c52f53c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-29 16:54:48.000000000","message":"Patch Set 21:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/21/check/tempest-dsvm-full-focal-py3/c00f0c0/ : SUCCESS in 2h 07m 03s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/21/check/devstack-plugin-ceph-dsvm-focal-py3/8ae0e5d/ : TIMED_OUT in 3h 14m 47s","accounts_in_message":[],"_revision_number":21},{"id":"4b95c544dfa495f29f7bbae7d4c2a8a1554fbac6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-29 16:57:02.000000000","message":"Patch Set 21:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-21/ns5_nfs : SUCCESS in 3h 25m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-21/ns5_iscsi : SUCCESS in 3h 22m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":21},{"id":"ddbcbf389ade6b6f72a22b27d37ce23e54792f13","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-29 17:34:57.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/21/check/iscsi-driver-nimble/2db21ca/2db21ca.html : FAILURE in 1h 37m 01s","accounts_in_message":[],"_revision_number":21},{"id":"d545382c0728da9eb89f574d9a52fd2383422e5e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-29 18:48:50.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"5e6add0e8e61b36c91001bc62a5c73ac1bbd6620","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-04-29 18:49:07.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"fb137c7bc9cf45d4eb5dfcf5e09d77d7dbf7a94c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-29 19:07:25.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"e86458f443e07132b313f18f979a0762a9b9965b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-29 19:07:26.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"94c4fc62c6a6be22a0eee2f8e25fe90e7613e8b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:07:27.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"9e3f0c475d1ac27c202e2d03da46e344ed1c02b3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-29 19:07:28.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"ce7e966bce12f72d1dfbba44d63870fb422508ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:07:39.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"f73dd6200ff6139fa9f0ff954fe0f1f76c4a5d35","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-29 19:08:04.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"69c842ea7a34b419d20dff5f7197981abbdfd8e1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-29 19:59:38.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/21/iscsi/ : SUCCESS in 1h 10m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/21/smb/ : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":21},{"id":"cbf8664f75e32d9fad06ac4e44228b112d21435a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-30 10:12:11.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/750782-19/Lenovo-iSCSI/ : SUCCESS in 3:37:52\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/750782-19/Lenovo-FC/ : FAILURE in 4:28:43","accounts_in_message":[],"_revision_number":19},{"id":"d6fe7b7c4e93406d5271cf96238389f359dfaa0c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-30 20:01:22.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/21/20220501023929/NEC-V-FC : SUCCESS in 1h 51m 56s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/21/20220501023929/NEC-V-ISCSI : SUCCESS in 2h 21m 50s","accounts_in_message":[],"_revision_number":21},{"id":"d35a4600c4fd9a7930e4b78d39a68a20eaa1679c","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-05-02 13:48:53.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4128628ce0ce5cff2022b2afbe570640a0727a7c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-05-02 13:49:12.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"571b6cc37d8d770f5a60b94d7d8d89c014ffbcf6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-02 13:52:05.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/21/20220502193433/NEC-FC : SUCCESS in 3h 16m 27s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/21/20220502193433/NEC-ISCSI : SUCCESS in 3h 17m 28s","accounts_in_message":[],"_revision_number":21},{"id":"6621780484dcf35a07adc7062821c0b0d0499ec2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 14:12:00.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41242ce9e6044970a4a6a066fecda557 : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eeda2bbe759f48fe9ae027bde7c4173b : SUCCESS in 20m 15s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"1db7f930d737eff00b5bb026a8ba686f891216f0","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-05-02 15:06:50.000000000","message":"Patch Set 21:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/7858bee/index.html : SUCCESS in 1h 16m 21s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/bdc5351/index.html : SUCCESS in 1h 15m 09s","accounts_in_message":[],"_revision_number":21},{"id":"0c1bf60cf1202012ed6b5f64a7bdd1e42e2995a6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-02 15:10:07.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/21/iscsi/ : SUCCESS in 1h 20m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/21/smb/ : SUCCESS in 1h 16m 09s","accounts_in_message":[],"_revision_number":21},{"id":"614878ffa0b5d8e2832ed8858ae995f71807af7e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-02 16:20:48.000000000","message":"Patch Set 21:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/21/check/tempest-dsvm-full-focal-py3/f00da70/ : SUCCESS in 1h 39m 43s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/21/check/devstack-plugin-ceph-dsvm-focal-py3/9be5e24/ : FAILURE in 2h 30m 09s","accounts_in_message":[],"_revision_number":21},{"id":"28238b555f5227cc03a9b1811cdd153ee793bd83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:26:42.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe569302e5c4b9f9d7ae4ee609b9723 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec8220faab0d4c698434f491b48799ba : SUCCESS in 10m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66be39ab88f445f599cf3c0aa5810fe0 : SUCCESS in 11m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e63fe55525284189b5e061dad57a117a : RETRY_LIMIT in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3861a9c442498283fac33a3e9e49dc : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/b41a0796e07b41ae87bfadb1ff50e931 : SUCCESS in 1h 13m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7aa0b863dcf4180833e5b05c731399a : SUCCESS in 1h 17m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54c85604bfde461c99380dd1e21f79aa : SUCCESS in 1h 05m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f86209cf025943f79182346eb6cec594 : SUCCESS in 8m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9dfc98bb7c814982a4a1442a8bf32642 : SUCCESS in 18m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f8719b2d65894bb2bd66a2e71d5b58c1 : SUCCESS in 5m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/764716f0e5424f78827670e2c97c8587 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b751ef88d164485385da647afdccad50 : SUCCESS in 10m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/035b3b8b89644a7c8318fef419bdef06 : SUCCESS in 10m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9776458e1c7843b5bc14e35791b1ff29 : SUCCESS in 45m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c7f0891b8e1d46dc99645fc39e3553da : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8b355e4b7644d98b81f97e56741af50 : SUCCESS in 1h 41m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/81a0657061404a639f23a2c0edc01702 : SUCCESS in 1h 15m 45s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/2daa06bc919a4dd7bb15e16567e8e502 : SUCCESS in 1h 16m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df457c35427a43b199ec186d0f5218f3 : SUCCESS in 1h 57m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/450f2fbd30e843c78ee9e40a425e94bd : SUCCESS in 1h 14m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9156d022544c49419b5c9250c2e50641 : SUCCESS in 1h 28m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f8a940c52fbb4676b349a0f8040e4ba0 : SUCCESS in 1h 03m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8f02bb42f779404b8586e45dc44242ba : SUCCESS in 1h 32m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67387b4b274f4b938e9c236b7a340c81 : SUCCESS in 2h 31m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd807e8734e9403391135e501dad8b09 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":21},{"id":"f1113af72eecfd72416462ff214ee1fe2fcdca38","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-05-02 16:45:31.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-82-750782-21/1651508802 : SUCCESS in 18m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":21},{"id":"607127dbe71d2515618b512407c1b8fc4f31d7a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-02 17:02:40.000000000","message":"Patch Set 21:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/02_05_2022_16_26_48\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":21},{"id":"0849bb1e45531268e0b486603a28d500143c33ee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-02 17:42:22.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_PowerFlex/9414282/DellEMC_PowerFlex/682 : SUCCESS in 1h 14m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"f9678459b1a894b17a6eb5faf9644a55f091f9d1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-02 18:19:12.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_SC_FC/914a27b/DellEMC_SC_FC/554 : SUCCESS in 1h 00m 22s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_SC_iSCSI/f4460b0/DellEMC_SC_iSCSI/575 : SUCCESS in 1h 52m 09s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":21},{"id":"6d886cbc4267159b74483b37a032e30f681cf545","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-02 18:37:37.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/750782-21/HPMSA-iSCSI/ : SUCCESS in 53:41.19\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/750782-21/HPMSA-FC/ : SUCCESS in 44:57.92","accounts_in_message":[],"_revision_number":21},{"id":"acf53485dc7cd96a9706f0dc3aaab50565524049","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-02 19:08:15.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_VNX_iSCSI/06e7fbb/DellEMC_VNX_iSCSI/343 : SUCCESS in 2h 41m 08s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_VNX_FC/4b61cc3/DellEMC_VNX_FC/359 : FAILURE in 9m 31s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":21},{"id":"18b570364ce69b4a7da5a6ad6ce01cdb6c4c899c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-02 19:12:50.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/750782-21/PVME-iSCSI/ : SUCCESS in 1:05:47\n- PVME-FC https://os-logs.seagate.com/dell-ci/750782-21/PVME-FC/ : SUCCESS in 1:04:57","accounts_in_message":[],"_revision_number":21},{"id":"829ef240d0e8b9a971c69cf1ec39d97525fa87d0","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-02 19:14:36.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/750782-21/Seagate-iSCSI/ : SUCCESS in 1:10:05\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/750782-21/Seagate-FC/ : SUCCESS in 1:05:24","accounts_in_message":[],"_revision_number":21},{"id":"f0ac16cf2c0c70a2d96d9e59f5512aaa6c5f8f0c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-02 20:17:51.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 3h 27m 40s","accounts_in_message":[],"_revision_number":21},{"id":"84f1e9261d8c3e1128c6e70dc7e319cd00dec61b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-02 20:56:44.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_Unity_FC/f9e857c/DellEMC_Unity_FC/715 : ABORTED in 3h 00m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_Unity_iSCSI/48c80b4/DellEMC_Unity_iSCSI/625 : SUCCESS in 1h 50m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"5f400888a2509fbce097bfdd7318efe3130fa9a2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-05-02 21:49:01.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_PowerMAX_FC/7d505b2/DellEMC_PowerMAX_FC/505 : FAILURE in 4h 58m 57s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_PowerMAX_iSCSI/02d5648/DellEMC_PowerMAX_iSCSI/510 : FAILURE in 5h 21m 56s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":21},{"id":"9d9c34654140b4d3860531999e36e7dbf260fdcc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-05 08:32:33.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/53334f00b6c845b0b93426e601b78310 : SUCCESS in 1h 37m 46s","accounts_in_message":[],"_revision_number":21},{"id":"7b5d7c3d6324b56f574377285b788eef2a835d9c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-05 15:49:03.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_PowerStore_iSCSI/b390d9e/DellEMC_PowerStore_iSCSI/472 : SUCCESS in 1h 18m 29s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/750782/21/check/DellEMC_PowerStore_FC/47ca316/DellEMC_PowerStore_FC/492 : SUCCESS in 1h 14m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":21},{"id":"71aa49bc498521de7665d183937e79ceca1f7b90","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-06 02:14:06.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/750782-21/Lenovo-iSCSI/ : SUCCESS in 1:06:32\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/750782-21/Lenovo-FC/ : SUCCESS in 1:01:09","accounts_in_message":[],"_revision_number":21},{"id":"b650ce54eaa5223eab7b9193a57dfe223e6b03a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-05-16 16:58:49.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"c3240cd615f91da7589b4212c1fab98dae731f99","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-16 17:02:58.000000000","message":"Patch Set 22:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":22},{"id":"bf453a37e8f0686a6c6d3edcd473dcddde4536c4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-16 17:11:17.000000000","message":"Patch Set 22:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":22},{"id":"478c09a674fb7f7d5a6116b8b3b3e15d9e2925ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 17:23:34.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d8c3f6e4ca6474398a46ade5a59047d : SUCCESS in 22m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5979512740743b59109e44b74a65244 : SUCCESS in 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"578cece6dbe99c5e75d38394d4bae6c108dce3c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-16 17:55:46.000000000","message":"Patch Set 22:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652720338603 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"e472bcba8ee8c04a50f6d22586d71353c8f40d1a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-16 19:01:33.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/22/iscsi/ : SUCCESS in 1h 21m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/22/smb/ : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":22},{"id":"4e988d311eb70f897fdb333033880edcc5f74f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 19:23:29.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84785352ea114ea7b5ec384ce6175a2a : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed0d836a7bfa4ed297a6241a44db43e0 : SUCCESS in 11m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca257ab75ce54622b4ae6d8623c29d6f : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38bb48b52ee74370938733e23ee8f64f : SUCCESS in 12m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2efd73640c2542779c813cad61a1f11b : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/1232b8fae7a0425ab55354e942d7d8cb : SUCCESS in 53m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0471d844a28b4b269095936216b64f9d : SUCCESS in 1h 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da1754d59e404a1c893d40eb99bad321 : SUCCESS in 46m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05ecc4dea5bd441da343f69ae8abd5b1 : SUCCESS in 8m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d70a7c7398a94dd48550f7ce838a295e : SUCCESS in 16m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/281190d313064a568ccb096946c341d4 : SUCCESS in 6m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ca6f7f946a3947c19567257fe3cf47c7 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4616e86a67e04eef8adc003040df65ac : SUCCESS in 7m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b62ee4968104d66a3bd09cd6d2894f5 : SUCCESS in 7m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/716365e704424a3d8f693970ce4439ac : SUCCESS in 38m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38bee690a6c149578189a4127e0c637c : FAILURE in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9cde16342bc348b086cc9d769a53df9c : FAILURE in 1h 23m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e0910bc7dc54424cbe981b90f9ba2fa1 : FAILURE in 1h 19m 11s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/c7c283e29bce4218800c4b96ab2b16d4 : FAILURE in 2h 06m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a53ab55333834adaa53109dd60e4b77c : SUCCESS in 1h 51m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/09787711de15439e83d45f8f442abb23 : SUCCESS in 1h 21m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/65c62b0bf606408ebdcf215ba88a8a57 : SUCCESS in 1h 16m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/082aac5917784265880266c8b3220884 : SUCCESS in 44m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5530e404d0b745b4ab41bef529b5ed53 : SUCCESS in 1h 12m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2ee24446c2440fc8f1c999888040bcf : SUCCESS in 2h 21m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0bfbc15eef14a9eb965b63d391c65f5 : SUCCESS in 49m 27s","accounts_in_message":[],"_revision_number":22},{"id":"01878406edbc3d783f7b636a3fbece1c0ec5e69a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-16 20:54:32.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":22},{"id":"50722ed90dc35b51e2ce9f0940cfea3705585880","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-16 21:01:43.000000000","message":"Patch Set 22:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/659cb8379f374ee69012b499e5d8aedc : POST_FAILURE in 2h 47m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ef72ac14b4df4b1aa144d8b358e9b99d : TIMED_OUT in 3h 09m 17s","accounts_in_message":[],"_revision_number":22},{"id":"2d353157f68fb1d00b8a875b0872ad10eb60013d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-16 21:36:43.000000000","message":"Patch Set 22:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-22/ns5_nfs : SUCCESS in 3h 43m 5s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-22/ns5_iscsi : SUCCESS in 3h 53m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":22},{"id":"eb9b2683629aa143df671811667d1efe8d51f517","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-16 23:29:36.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/22/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/de16ea5/ : TIMED_OUT in 3h 03m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/22/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/24ec732/ : SUCCESS in 1h 11m 07s","accounts_in_message":[],"_revision_number":22},{"id":"3f69e343e08c06fbaf1a96a8b32701ad958c36c6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-17 02:02:33.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/22/check/iscsi-driver-nimble/abefccd/abefccd.html : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":22},{"id":"b8dc19eb7d93d458154e17a200b9e6307369acbf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-17 03:57:54.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/22/check/DellEMC_Unity_FC/56ba69d/DellEMC_Unity_FC/758 : FAILURE in 1h 16m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/22/check/DellEMC_Unity_iSCSI/d30bc6d/DellEMC_Unity_iSCSI/728 : SUCCESS in 1h 44m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"3ccf3ffcc5e14b8ff969842e7fbca49d4f7ac1d7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-18 20:37:32.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/22/20220519021709/NEC-FC : SUCCESS in 3h 18m 11s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/22/20220519021709/NEC-ISCSI : SUCCESS in 3h 20m 21s","accounts_in_message":[],"_revision_number":22},{"id":"62c5586e78db1dd430ad15470cdcccca4faf2686","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-19 02:25:59.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/22/20220519055250/NEC-V-FC : SUCCESS in 5h 14m 07s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/22/20220519055250/NEC-V-ISCSI : SUCCESS in 5h 33m 06s","accounts_in_message":[],"_revision_number":22},{"id":"42d35965221c5ad66239b4c85c9d992cdbf31865","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-05-23 17:09:37.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"e712035b1a981bfc98bb1408cca3d77e808fa688","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-23 17:12:08.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":23},{"id":"b99ca350beba51b5bdef8522bb453423e51ab7fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 17:35:24.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09708e3988ae40f0a91a0ee64e47778c : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb8a30276fe3493fb6bc12b1a513e559 : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"a38084aab2664d64fbe695a65894a364d056dc52","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-23 18:03:52.000000000","message":"Patch Set 23:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653325786430 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"82a1c609b2f05131dce849f1dcf59eeac5b747fc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-23 18:36:56.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/750782/23/iscsi/ : SUCCESS in 1h 17m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/750782/23/smb/ : SUCCESS in 1h 26m 46s","accounts_in_message":[],"_revision_number":23},{"id":"83744aee14128edec96f4ac6faa940f0bf7428bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 19:24:46.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f53c038ef0b4a6b80aaa15b01a17277 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76621d1e0d3b497f838cb9e57443ce1b : SUCCESS in 10m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/917130a261614bfa8c3b2b502b2d4573 : SUCCESS in 10m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f145defc39c74fefb7d73b9eb7c815ba : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833ae387edbf4ef29b69f02779be3c72 : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/45981e999da4446caad59f2752a31ceb : SUCCESS in 54m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b66491a5b9d443687b471163eeabe13 : SUCCESS in 1h 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7f15cbde17d477e91641ab1f5d969ec : SUCCESS in 51m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b195ea74599a4ff193f81570e6721d46 : SUCCESS in 8m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dfde7f5f575a457099d479300daed13c : SUCCESS in 15m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a3ee6a3622ed4d8d83e0962610bbbee8 : SUCCESS in 6m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93edc8ba32f945b7b5ee202a052fa9bd : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/15ec6d112cf2466ebe3df271d2a70de9 : SUCCESS in 6m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/26b783c1c51544628ce09085849eb3a8 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9b4c7d585dc54ff2a842a0bbe2c93f09 : SUCCESS in 42m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b7f6321f447e45dda3eeaf7b19cc4723 : FAILURE in 7m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59de0ec7372548c8a19084a4ef407da6 : SUCCESS in 1h 12m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/de75bab4ed874a5581a01d492f22b7db : FAILURE in 1h 18m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d156b30fb59e481c95fddd7277aaedb0 : SUCCESS in 1h 28m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/810e69547f4a439eac9649c37d6988aa : SUCCESS in 1h 14m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/99c9759c41884db9aa23e141b2f55246 : SUCCESS in 1h 17m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/45492d57f8854bc69a405a0b2266d854 : SUCCESS in 42m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2bec67cb343a45c28da0930bec7df299 : SUCCESS in 1h 33m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36e8e64f541d43b183ae8292bf8df861 : SUCCESS in 2h 12m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e52530b084d6421a92ad40cc2157af4b : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":23},{"id":"845940a007bc98e3afc3bd8f4c9e00986688e537","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-23 19:24:53.000000000","message":"Patch Set 23:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":23},{"id":"e399d14d76a729d60711babd3b0b582dfad40137","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-23 19:35:53.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_VNX_iSCSI/dd5a4d2/DellEMC_VNX_iSCSI/440 : FAILURE in 1s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_VNX_FC/d2722b5/DellEMC_VNX_FC/459 : FAILURE in 10m 46s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":23},{"id":"69d0624afe49e89a180ad919c25a69fa0b7f0fdb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-23 20:24:54.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/84f006ae4342419ca66b12d51b22f7c9 : RETRY_LIMIT in 11m 26s","accounts_in_message":[],"_revision_number":23},{"id":"ae23e53f01339ded6bd4ae6a6ade6d67ec4d08a2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-23 20:26:10.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/23/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/353a28a/ : SUCCESS in 51m 39s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":23},{"id":"9a664c5502f916e361f9a0e7ba744b897e54c4c5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-23 20:29:32.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/23/upstream-check/cinder-tempest-plugin-ontap-fcp/e5f61a3/ : SUCCESS in 1h 03m 26s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/23/upstream-check/cinder-tempest-plugin-ontap-iscsi/9f0e17a/ : SUCCESS in 45m 45s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/23/upstream-check/cinder-tempest-plugin-ontap-nfs/1973f37/ : FAILURE in 44m 07s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"74b46966d52d11dd224c6e7f2d69328d1430fb2b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-23 20:49:29.000000000","message":"Patch Set 23:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-23/ns5_nfs : SUCCESS in 3h 34m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-23/ns5_iscsi : SUCCESS in 3h 24m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":23},{"id":"da7e07a11f16d823655d770ee2bad2d14d6dbe56","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-23 21:08:22.000000000","message":"Patch Set 23:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/23/check/tempest-dsvm-full-focal-py3/d4133ac/ : FAILURE in 2h 37m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/750782/23/check/devstack-plugin-ceph-dsvm-focal-py3/490bdb3/ : SUCCESS in 2h 54m 56s","accounts_in_message":[],"_revision_number":23},{"id":"cb3a66a068f24efb25a94fd7a3e19e1d27189f78","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-23 23:51:10.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/23/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a7908e8/ : TIMED_OUT in 3h 03m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/23/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1d48d41/ : FAILURE in 1h 23m 37s","accounts_in_message":[],"_revision_number":23},{"id":"ba085fe6f28289528bfce4c1628cd51d2e468541","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-24 00:23:21.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_PowerFlex/1c4bd48/DellEMC_PowerFlex/783 : SUCCESS in 1h 07m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"ab16cee5738a9451297484c57d95ae9bfabd0cb7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-24 01:31:45.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/23/check/iscsi-driver-nimble/8ff8db0/8ff8db0.html : SUCCESS in 1h 49m 26s","accounts_in_message":[],"_revision_number":23},{"id":"9500b20e04a7eee8b75cdfbd31d0ad500c06642f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-24 02:12:51.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_PowerStore_iSCSI/f7ea736/DellEMC_PowerStore_iSCSI/528 : SUCCESS in 1h 23m 13s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_PowerStore_FC/62dd685/DellEMC_PowerStore_FC/553 : SUCCESS in 1h 09m 11s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":23},{"id":"cf13c5e8570a649fa2d4741368eb7a0224cf9856","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-24 05:07:05.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_SC_FC/acae5e1/DellEMC_SC_FC/630 : SUCCESS in 48m 04s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_SC_iSCSI/bfd9c86/DellEMC_SC_iSCSI/651 : SUCCESS in 1h 45m 17s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":23},{"id":"823e16112231289d7526bd7651257f57d4fde109","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-24 07:05:06.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/750782-23/HPMSA-iSCSI/ : SUCCESS in 54:16.67\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/750782-23/HPMSA-FC/ : SUCCESS in 47:06.81","accounts_in_message":[],"_revision_number":23},{"id":"4baf238da1ebc24acc1fb18604191108cb507d6e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-24 10:57:39.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/23/20220524173514/NEC-V-FC : SUCCESS in 2h 12m 34s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/23/20220524173514/NEC-V-ISCSI : SUCCESS in 2h 22m 22s","accounts_in_message":[],"_revision_number":23},{"id":"06c869007412ad1d4c6a7221fa9d88704dc8a8c6","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-24 11:56:55.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/750782-23/PVME-iSCSI/ : SUCCESS in 1:02:11\n- PVME-FC https://os-logs.seagate.com/dell-ci/750782-23/PVME-FC/ : SUCCESS in 1:01:58","accounts_in_message":[],"_revision_number":23},{"id":"3b07d781949d10d04080fa0bdbdb029d4b91cba4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-24 12:10:00.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_Unity_FC/6820fc8/DellEMC_Unity_FC/823 : FAILURE in 1h 19m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/23/check/DellEMC_Unity_iSCSI/3e94a5d/DellEMC_Unity_iSCSI/791 : SUCCESS in 1h 42m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"ce4b18baefdc840445034e77f617fd2bef5bc059","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-05-24 12:44:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":23},{"id":"2b119a898c4f549a1c2db355d0d668f5fdb759b8","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-24 13:46:30.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/750782-23/Seagate-iSCSI/ : SUCCESS in 1:13:39\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/750782-23/Seagate-FC/ : SUCCESS in 1:14:28","accounts_in_message":[],"_revision_number":23},{"id":"184a0d6cb14da6419789e3a1c3ad12a06c1f0622","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-25 08:38:22.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/750782-23/Lenovo-iSCSI/ : SUCCESS in 1:05:20\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/750782-23/Lenovo-FC/ : SUCCESS in 1:10:16","accounts_in_message":[],"_revision_number":23},{"id":"0c7caed2371717ff6592e90c61a10a260231d0f9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-26 23:13:10.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/23/20220527015450/NEC-ISCSI : SUCCESS in 3h 22m 38s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/23/20220527015450/NEC-FC : SUCCESS in 6h 18m 17s","accounts_in_message":[],"_revision_number":23},{"id":"1db0e5790139f157b59707464451f7d21a575115","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-29 09:30:38.000000000","message":"Patch Set 23: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"af11fab34e206e48b2cabd5855e0b80185fa4f13","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-10-27 18:44:30.000000000","message":"Patch Set 23: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"89db301c0b6ac6a468abd2e638040c7c1c565696","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-10-27 18:44:36.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":23},{"id":"dd15aa6fe6f5ff57004e7d9eee34fbf417bc7ea4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-15 11:29:02.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":24},{"id":"cc5848b11ee71535ab6327cf0a9173cfe1e87288","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-15 11:29:13.000000000","message":"Patch Set 24:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1730 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"e40afc4a9a3e29952afe8ff805346a55ef6e9324","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-15 11:29:33.000000000","message":"Patch Set 24:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":24},{"id":"62bee5e6ec41b8709dec28976920b3c78e6ec69f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 11:48:31.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69541ae1a17a439c88cef0086ed71183\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1611cc1e11d14781b42296f357c793cf : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a072b2b753f40f8b3854ce0096d2a4a : SUCCESS in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"9d024ad3b6894aaede4e6e6a476222cb5f061248","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-15 11:55:43.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 15m 03s","accounts_in_message":[],"_revision_number":24},{"id":"c35ee166e3b62535e70fc20711f245b9067df599","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-15 12:44:32.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/24/ : FAILURE in 1h 03m 00s","accounts_in_message":[],"_revision_number":24},{"id":"2b67864bd2d2f19f4e6dc1885827a8e6011fc9ee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-15 13:12:59.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/24/check/iscsi-driver-alletra-6k/a32d8d2/a32d8d2.html : FAILURE in 1h 40m 30s","accounts_in_message":[],"_revision_number":24},{"id":"1eabb39a33eddbacd712ac01b42b54239276a1da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 13:13:58.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a09594950041829cf4163310c13225\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ec7322d2384b57a4b4d99b8a1327f5 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/408df5cb7ee641eeaaae1afb425d80f2 : SUCCESS in 10m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5d6577da9234c17af8b495340ca11e2 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb89317a85c47b9bf70ae1547c223fb : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/02649b198f2345e9a8a8563a596906d6 : SUCCESS in 48m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/65251f72ad97465b8f59f87c2950c2b6 : SUCCESS in 38m 40s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7073bb149db4327ab1d7e9d7ff8c660 : SUCCESS in 1h 23m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8ef58d3e8de48c189b1ae7d4613094d : SUCCESS in 46m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e00413b6595e42f38b8ce5a187cbe31e : SUCCESS in 7m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/499196343645455fa7446a09b519d645 : SUCCESS in 11m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/76220462d5264beca1a5a7c157955086 : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0861c48533f44772a5190ad96600ebff : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/02fcc542208a4cdc95afab69b88ecd3b : SUCCESS in 5m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/70e0af11aac64afd9b93734b890b1dbc : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf7cdeac7c334103a368d3c56df70bbc : SUCCESS in 24m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/724f926ebdc34e3db78d6a83b8a3455b : FAILURE in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/44d81efddd3f43e6a82c68250714ac90 : SUCCESS in 1h 04m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5addf64e11dd4952ad6805e6fe0e6417 : FAILURE in 1h 20m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee07101783c64d14b71cc55767484a49 : SUCCESS in 1h 16m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/98c172262cf6477ab7063c9c88864e4d : SUCCESS in 56m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/92680eb615d74e0b94f7cf8d7d5aad80 : SUCCESS in 1h 11m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b8c3ca0b1fda4b2680f997c5f01d8a2c : SUCCESS in 36m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0f629379d22a4e659d647d859d572f91 : SUCCESS in 1h 13m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1081f6bd32b541a0bdeea10511d28976 : SUCCESS in 1h 43m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca8d7a49f66d41eca97a23c595168395 : SUCCESS in 47m 14s","accounts_in_message":[],"_revision_number":24},{"id":"ead5eddce4289eb1e6bc4c1d750726f8d875fbee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-15 13:23:13.000000000","message":"Patch Set 24:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_11_2022_13_14_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":24},{"id":"ed922bb9b0763f36a038807ad09fb79f3535f5bf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-15 13:30:42.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/24/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/07da669/ : SUCCESS in 1h 52m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/24/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f8a6933/ : SUCCESS in 1h 44m 24s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/750782/24/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/469782c/ : FAILURE in 2h 00m 32s","accounts_in_message":[],"_revision_number":24},{"id":"b6121a2e1c667d3ecd5c4df67f14df2609f63f4f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-15 13:51:21.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 2h 18m 43s","accounts_in_message":[],"_revision_number":24},{"id":"1004448e15bfc9dc1c0b90f8d8c85331300f0568","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-15 13:53:09.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : SUCCESS in 2h 19m 04s","accounts_in_message":[],"_revision_number":24},{"id":"bc23a12e81133ac7dd9b5ea0482d11399a5070a1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-15 13:59:20.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/750782/24/check/iscsi-driver-nimble/15daaa3/15daaa3.html : FAILURE in 2h 05m 31s","accounts_in_message":[],"_revision_number":24},{"id":"b685fad6a95fe9f5214d1f8a3d1266b03cbe9fda","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-15 14:02:54.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/24/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/617a100/ : FAILURE in 47m 27s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":24},{"id":"9a35bd1dc9d306b2b4970571aa7970ad531ba0fa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-15 14:29:05.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/24/upstream-check/cinder-tempest-plugin-ontap-fcp/f921db8/ : RETRY_LIMIT in 7m 59s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/24/upstream-check/cinder-tempest-plugin-ontap-iscsi/d3ec8aa/ : FAILURE in 48m 19s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/24/upstream-check/cinder-tempest-plugin-ontap-nfs/3dccd9d/ : SUCCESS in 1h 13m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":24},{"id":"f7c0b88bdef22ad5e7de2c7cc89816cd2067dc9d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-15 14:57:43.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/24/check/DellEMC_Unity_FC/5d620a8/DellEMC_Unity_FC/1997 : SUCCESS in 57m 07s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/24/check/DellEMC_Unity_iSCSI/31f16ff/DellEMC_Unity_iSCSI/1844 : SUCCESS in 2h 00m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"fc5ddcccf118d87100877d15fb98e1d401459822","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-15 15:00:10.000000000","message":"Patch Set 24:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-24/ns5_nfs : SUCCESS in 3h 27m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-24/ns5_iscsi : SUCCESS in 3h 30m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":24},{"id":"108329638f51931bc2a6569cca296a01bda02e8b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-15 20:54:11.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/24/20221116021900/NEC-FC : SUCCESS in 3h 27m 09s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/24/20221116021900/NEC-ISCSI : SUCCESS in 3h 35m 08s","accounts_in_message":[],"_revision_number":24},{"id":"3cbdcefb01e497e16ad0aae7badaebc714b02ef0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-11-15 22:05:17.000000000","message":"Patch Set 24:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/825","accounts_in_message":[],"_revision_number":24},{"id":"9fd5af90eb6b1f2998fff5b740d3380cf898175f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-16 05:03:33.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/24/20221116073744/NEC-V-FC : SUCCESS in 5h 48m 10s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/24/20221116073744/NEC-V-ISCSI : SUCCESS in 6h 25m 46s","accounts_in_message":[],"_revision_number":24},{"id":"faee7370ad8ce858c3798151b76cac015235f7a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-22 16:00:41.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"aa2fc1b54f4ab405f8c027f6c26f36da6aa2cbd7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-22 16:00:51.000000000","message":"Patch Set 25:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1768 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"d8617a7a873f7fad9502bb0fd1949e49e00b27f8","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-22 16:01:14.000000000","message":"Patch Set 25:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":25},{"id":"258b67128cae527dcb3e8b5204c62241e078876a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-22 16:05:55.000000000","message":"Patch Set 25:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":25},{"id":"e6e22648a204137ebe334efa005bbd5d18f53ad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 16:28:11.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce5fd916cbbc4f8eafc672db8b8b10a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a963e8bdc0d43dabd51f3c6c221730d : FAILURE in 23m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ab6ed61dc2e477c9a1645084d810323 : FAILURE in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"c5eb2b99a7089cd2a7b6e2c81815ae4abd93564d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-22 17:40:09.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : SUCCESS in 1h 36m 31s","accounts_in_message":[],"_revision_number":25},{"id":"732ae4bd9ea16173230223bbad63135c5503d218","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-22 17:45:48.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/25/check/iscsi-driver-alletra-6k/6f06526/6f06526.html : FAILURE in 1h 41m 52s","accounts_in_message":[],"_revision_number":25},{"id":"0afe5782e6fe6c6ee8a6ac30ceefd7255c95d1fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 17:46:31.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4641d34d2dba434eb25760b0139dae46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0a73ce0c454ddfbb042c8988970b77 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/761c59538472410199fceb2d24a2ab93 : FAILURE in 10m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a9e9a424c8f42a7af4fe1e07e5fd681 : FAILURE in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c135ec9fbe44d1bbc57d8174c0f7d5c : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/7a489bcb9d444bcdb7550b61776b5de8 : SUCCESS in 44m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/53940fd3707b4c74a264790553529229 : SUCCESS in 56m 50s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40bdd06a318f41f5a1b954e7b1a13b0f : SUCCESS in 1h 08m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69c661c730dd4907b46eeffd6c63e209 : SUCCESS in 53m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a758f28d55f845f7808f35a259ff5c39 : SUCCESS in 8m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f97b918ef99945b2aab43ce9eed10340 : FAILURE in 12m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e3d55f2c31384dbb8efabc37203e232c : SUCCESS in 6m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/210e752d27b2428a8fdb1cf4fc70f2c0 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/96cf33f6bfcc4f4295a27b0bd9419ad2 : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36cfe8135c08457095e927f77205f6bd : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bb06fece4cb3490ab91323d173af82e7 : SUCCESS in 41m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d5e6154cdfe84558b6c247904af5ff31 : FAILURE in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/085fa1d8bf2c4608be82a98e3091d725 : SUCCESS in 1h 06m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/18bbd88127ff491eadffa3b9b47124f5 : FAILURE in 1h 23m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e5e062899a964eadb8dff519634de40d : SUCCESS in 1h 41m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b4232c81733c4091b367c1d3ef27b5f0 : SUCCESS in 1h 03m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac6e3a0d1858478d98228df4827583de : SUCCESS in 1h 21m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/831a4bdbe87c401d9a70a98ef14278b1 : SUCCESS in 31m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c033e90cb9540afa8ef8b0c80fe4deb : SUCCESS in 1h 23m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6df6fbe510d9464083fe9b781728ae22 : SUCCESS in 1h 34m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63c299a66522444c858d9131155ef366 : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":25},{"id":"8c49a1f871ec54083d435b374d34fceede04b3b4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-22 17:52:22.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 47m 48s","accounts_in_message":[],"_revision_number":25},{"id":"05535cf9e9145ffc17a0214c2607b344d9a4c0d2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-22 17:55:17.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/25/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/16b79ea/ : SUCCESS in 1h 44m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/25/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0890747/ : SUCCESS in 1h 53m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/750782/25/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/38b65ee/ : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":25},{"id":"09fefb44431bc38ea6524f4e9de9b76407792dcd","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-22 17:58:15.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/25/ : SUCCESS in 1h 53m 20s","accounts_in_message":[],"_revision_number":25},{"id":"7feb837e3fbd9f34e4e40902f471b54abeadc4cc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-22 18:10:49.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 29s","accounts_in_message":[],"_revision_number":25},{"id":"dad737b60b45c012b8a0c6cd302b547996bd9686","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-22 18:43:01.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/25/check/DellEMC_Unity_FC/25d9d5e/DellEMC_Unity_FC/2023 : SUCCESS in 59m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/25/check/DellEMC_Unity_iSCSI/a7fe2dd/DellEMC_Unity_iSCSI/1870 : FAILURE in 2h 07m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"e5802b760e359369ed15678550421b3eb5b62c3c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-22 19:54:37.000000000","message":"Patch Set 25:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-25/ns5_nfs : SUCCESS in 3h 30m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-25/ns5_iscsi : SUCCESS in 3h 34m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":25},{"id":"10107e2d2235ab1f9cf558893fd95114593d29e8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-24 04:35:15.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/25/20221124100037/NEC-FC : SUCCESS in 3h 32m 50s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/25/20221124100037/NEC-ISCSI : SUCCESS in 3h 34m 35s","accounts_in_message":[],"_revision_number":25},{"id":"154dd1904a6ace5658db2e22ba5ed096481db261","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-24 13:18:57.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"0213c306eeb3cb062b9ab2a8df321a58b13c6585","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-24 13:19:06.000000000","message":"Patch Set 26:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1777 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"05a2f7776e72f2b23f0a666aed29759844957bff","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-24 13:19:19.000000000","message":"Patch Set 26:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":26},{"id":"290b1ddc2855d4c8bf29d6bfa5633a0e21b588d1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-24 13:22:49.000000000","message":"Patch Set 26:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":26},{"id":"43ddf15df100004de4f93aff8a4783798469b026","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 13:38:14.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25087bbd9d3f4e5ba8512d1aede1e36a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2b8e682a5ae4748ab05faad97b78c40 : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c585014722a04389ab71f5d7c3eaeb75 : SUCCESS in 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"f1967bacbc2c4835e9bbf5f8dd18c9550b785ef9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-24 13:59:09.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/25/20221124170938/NEC-V-ISCSI : SUCCESS in 2h 40m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/25/20221124170938/NEC-V-FC : SUCCESS in 5h 49m 28s","accounts_in_message":[],"_revision_number":25},{"id":"76e10fb20873919ea2b1440ceca062c5e8475d18","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-24 14:27:06.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/26/check/iscsi-driver-alletra-6k/cc98141/cc98141.html : FAILURE in 1h 02m 33s","accounts_in_message":[],"_revision_number":26},{"id":"617d647f02b92504baefba247e1e221fef880f79","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-24 15:01:42.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/26/ : SUCCESS in 1h 33m 57s","accounts_in_message":[],"_revision_number":26},{"id":"2b6b2629b0f5ea2d44f523df179dd1a610c3e750","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-24 15:10:12.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : FAILURE in 1h 45m 23s","accounts_in_message":[],"_revision_number":26},{"id":"822e3a71aca6b562f946218cc6c133cad34ec465","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-24 15:10:57.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/750782/26/check/dsvm-tempest-cisco-zonemanager-job/9fa16a0/ : RETRY_LIMIT in 8m 56s","accounts_in_message":[],"_revision_number":26},{"id":"af4edaffa2bebc9c05960683586d7ac9572e8721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 15:14:05.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/383f83ae027f49c281f2571439ac16c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edbf2ea4838b40f9a2e256a5f2f600a2 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fad9f3077f864e1e85e7d336390d68cd : SUCCESS in 10m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/901ccbf5f1a64413b1c3e227eed75f1c : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d461ca1d20a4432c851bf734d4716758 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/7a454da0851f4bfbb8317011789290a0 : SUCCESS in 52m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cacddac0a31c4ed2b375fcd4456fe46c : SUCCESS in 56m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24fb2f2498a24a719684f91fbb040741 : SUCCESS in 1h 22m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc81164532464053a5371525f2aaf583 : SUCCESS in 53m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e2e0be44d154d28aabb8c4e22480629 : SUCCESS in 8m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3294eb1c743e49cca69e1f6955b8e8eb : SUCCESS in 14m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4942abc6c91a4f239fdc21029f511719 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/485b2a5ce11742489891a02fcf54cc9c : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/faf4e38342704028820d9ee0fd70ceb3 : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a8d6f4e1ea93403bbdc3025e02e335f7 : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cfe5ff5cdcc0464da4bab56e20b55ce0 : SUCCESS in 39m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47ed08a9b48f4a9fae7fb3534f0e7548 : FAILURE in 7m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed891f01dde54df69411d9613b01a778 : SUCCESS in 1h 06m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5a5b13f0bb63477cbdcd416c74604323 : FAILURE in 1h 24m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/565859e012684185956ddd356a980d7b : SUCCESS in 1h 24m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0556db1fe7dd40b991a016e7e938092e : SUCCESS in 1h 02m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df4fb6294bf34f1b9b1604ed746140c7 : SUCCESS in 1h 23m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8e59b5b8b7d0458683e835a4a3ba2944 : SUCCESS in 36m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78b4b6cb92cb4a3ea8acfda1aa66afca : SUCCESS in 1h 24m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7590c72ff22d417d985c440e6d8479b4 : SUCCESS in 1h 52m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd54d44553c44c4d8e8d771337ca34d4 : SUCCESS in 44m 00s","accounts_in_message":[],"_revision_number":26},{"id":"19a54334f281afcea07c26cf1c1159e2f9ee1e3e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-24 15:19:05.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : SUCCESS in 1h 56m 44s","accounts_in_message":[],"_revision_number":26},{"id":"395058cca5604bfadf5ff3c504e9beab7bf864bd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-24 15:53:45.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/26/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/edda04d/ : SUCCESS in 1h 41m 19s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/26/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/647aac7/ : SUCCESS in 1h 32m 13s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/750782/26/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/1a0d677/ : FAILURE in 2h 33m 35s","accounts_in_message":[],"_revision_number":26},{"id":"12c4bba5e9e84771e80bd51cf70a33fd53ca6281","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-24 15:55:49.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"59ffd9c7f97951e5d72b0c5a8a4a03ec66cd1047","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-24 15:56:01.000000000","message":"Patch Set 27:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1788 : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"be89d260c09bca77b844af569c6765ce724f7e7a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-24 15:56:13.000000000","message":"Patch Set 27:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":27},{"id":"86c9464d3c15b721db1150d0e3fe3e0c6b47e1d1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-24 15:57:33.000000000","message":"Patch Set 27:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":27},{"id":"6f6abe58e8729920d28e13d953b75b52db5d282f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 16:22:45.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efa84942f157422fb6ca3553dd30608c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcb02e78cfc24e5f89d887ee09161eb2 : FAILURE in 20m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3b8a539c35a4e518ec04f0829ea073a : SUCCESS in 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"ea224786763ccd2af2f386906e0b45ee9ea9dc30","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-24 16:54:28.000000000","message":"Patch Set 26:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-26/ns5_nfs : SUCCESS in 3h 33m 44s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-26/ns5_iscsi : SUCCESS in 3h 35m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":26},{"id":"216af6d550f1443fc4fda0b0b11e8d559416ebeb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-24 17:44:50.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/750782/27/check/dsvm-tempest-cisco-zonemanager-job/eac21ee/ : POST_FAILURE in 12m 54s","accounts_in_message":[],"_revision_number":27},{"id":"776352af0b1e5053c7eaffc33e2caa6305c130e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 18:29:53.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cd244d6ff6c4a94a97f3c5fe51f373e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f399a439a047edbd7bed8c3dbd7aeb : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac295aa51e6644a685f2a074a4789a58 : SUCCESS in 9m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f72b14da386547898524358a49acab87 : SUCCESS in 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f031755dbfa4802bbfe0f2ca47fa35f : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/69878fc950934dc390896307b413ce74 : SUCCESS in 54m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5bae2777e5be4757977dd8bb12832fd6 : SUCCESS in 55m 39s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c08dfbecd5a24040a4f4d67b0daf2255 : SUCCESS in 1h 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/beef1ef251fc459789b7197feaaba528 : SUCCESS in 50m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d0c35c22ef9495f9b098b75569ed78c : SUCCESS in 7m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a4464a5ca13f4eeca391a9ea151447c3 : SUCCESS in 21m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/28b03c559ac843f3bc1e23eea8ddec14 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/866219b940934273a4820721f1779efa : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d1b73cef98be411abfecbe95da06363f : SUCCESS in 5m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be0060dcb61b4029b29959247be37a48 : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d65f0c6e0597452fa9a468029c61294d : SUCCESS in 35m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88295c5696754cae9e6c992a1adb67b5 : FAILURE in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58528ebca4414c47814a8887a6109b95 : SUCCESS in 1h 18m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dc090ee9f199491aaa606fbc7befd085 : FAILURE in 1h 25m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bd491e6a8c3746bfaa7e72a8193738d5 : SUCCESS in 1h 35m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e184376f87a8410ba93f9cf0d1a3a137 : FAILURE in 1h 16m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/214372c3df9c41b2ae37b9f7c06a872a : SUCCESS in 1h 41m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f628b6aff0845698aa049844f377684 : SUCCESS in 36m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b3bcf48772348c5b7f4514b2e221015 : SUCCESS in 1h 22m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/149af54abbdd4ed18721c69e559b82fb : SUCCESS in 2h 27m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6baf5231d0314b4aad3fc89fdc1be4af : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":27},{"id":"849db78e96e76dd0686c0bbb3e4533cbc9b9aa14","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-24 18:54:31.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/27/check/iscsi-driver-alletra-6k/068e7d5/068e7d5.html : FAILURE in 38m 37s","accounts_in_message":[],"_revision_number":27},{"id":"1115f52104aac7c15b5f2c607b42a1ffe602b16a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-24 21:18:01.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 51m 22s","accounts_in_message":[],"_revision_number":27},{"id":"d0ede6191520a4883e0957a8d8e70ffdd380cc70","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-24 21:35:30.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : FAILURE in 1h 50m 21s","accounts_in_message":[],"_revision_number":27},{"id":"4780939e4e6d4745373ed18aabab22acc7289d77","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-24 23:02:21.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/27/ : SUCCESS in 1h 50m 05s","accounts_in_message":[],"_revision_number":27},{"id":"acfdea3332454b3add538efe7d121cf52eac2e11","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-24 23:12:55.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/27/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2ac6321/ : SUCCESS in 1h 40m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/27/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/862652d/ : SUCCESS in 1h 36m 53s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/750782/27/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/769ac1c/ : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":27},{"id":"77af3cf834e6f09184bc82413fc326abec55f912","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-25 04:28:42.000000000","message":"Patch Set 27:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-27/ns5_nfs : FAILURE in 10h 43m 16s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-27/ns5_iscsi : SUCCESS in 3h 40m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":27},{"id":"5af9e0960d0da152011c677aa7ce447440bd3432","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-25 10:05:31.000000000","message":"Patch Set 27:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":27},{"id":"aca89ac82858dc27fcff7f887ad4d10bad6cae23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-25 10:05:42.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"62cf65eef1905e0e7611d9471fa4a0515999177a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-25 10:05:53.000000000","message":"Patch Set 28:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1795 : FAILURE","accounts_in_message":[],"_revision_number":28},{"id":"f2b1fd3d2b03f6a614de69e2c4c6a09ebeb5b3bc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-25 10:10:59.000000000","message":"Patch Set 28:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":28},{"id":"9c164062219cabc81eaa99e6f62c2b9cb9a53a3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 10:26:06.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a3839e9132e470e86de90e5f1206909\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81145b8760f34e56b8c4a34902e584c3 : SUCCESS in 14m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ab9fbffc6924e89b4795dee82801fd9 : SUCCESS in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"6c3b91a3fab6d82d7d09ddbb76e9cdd8d0dc5db2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-25 11:33:26.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : SUCCESS in 1h 24m 37s","accounts_in_message":[],"_revision_number":28},{"id":"3a814fada7dae0ccedb4f2ad7335aad4bb0c0140","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-25 11:38:01.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 29m 04s","accounts_in_message":[],"_revision_number":28},{"id":"932efc5bd85dd547b35d85b90df14b2505619e08","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-25 12:09:08.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/28/check/iscsi-driver-alletra-6k/88f1269/88f1269.html : FAILURE in 2h 00m 29s","accounts_in_message":[],"_revision_number":28},{"id":"1ed036d3b8927cd59e432c95a4202339650d45b2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-25 12:13:43.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/28/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cc838e6/ : SUCCESS in 2h 06m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/28/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8ac5861/ : SUCCESS in 2h 00m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/750782/28/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/9f1a714/ : SUCCESS in 1h 51m 10s","accounts_in_message":[],"_revision_number":28},{"id":"88b1c283b59e3b335cfcd8b5ccf2ef1c3a3d2572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 12:45:26.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b7dc80046ff484d9a4796238d6a89f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ae0e6b13534c8d8325fcf1af287a4e : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c1f05833d6347e5bc9e4000628b2122 : SUCCESS in 11m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d3db120f9dd490c9d6e9f42322173de : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178b64392c6e4ea0b328144df6da5ac5 : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/ec6833ab644d410b8f253468d4fc40d0 : SUCCESS in 54m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/42a61c5bee9442e2843d077611634bfe : SUCCESS in 1h 04m 27s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02ab15e4868f400a8cdd337fde5ea07c : SUCCESS in 1h 23m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9705022003af46ec8838a1129be5f278 : SUCCESS in 57m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb062bc15afa4195b00884753f524724 : SUCCESS in 8m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca12b3cd942f48b9973a658c00ccd272 : SUCCESS in 13m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/792a58f497a14ac782156658cb0eb6ef : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4caf136161914c66a2172af9235ab39a : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dcdbfeaeca834f59a068812f370623fe : SUCCESS in 6m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb3dd5ac13844c05afc40ef42d232bec : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e3ac23d11e254794a96ea021e0774497 : SUCCESS in 50m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bb8c4716c52a4ff9b530692f3d0f3dbb : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4eb48ab8f9814b5b89e914d5fdcccc5a : SUCCESS in 1h 07m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3db126d6a173449fbdbfe8ebf86e513a : FAILURE in 1h 32m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/81891fd27cdb44a1906c92854d415148 : SUCCESS in 1h 48m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4c3c03d525494e9f9bf803bee8ef1322 : SUCCESS in 1h 05m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b4b7dda5b6e4663844ded6ac7b1d07c : FAILURE in 1h 24m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/97fb4a736b53438bb88384a1be7e0041 : SUCCESS in 36m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/40d0dcb9f849409e90d0dbfbd1c5b309 : SUCCESS in 1h 20m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d876ef3b1b9243738564a3a2de04ff1c : SUCCESS in 2h 31m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bde04938f3f44ece9b9eea488845dbce : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":28},{"id":"803e0fb19b45e8317d3c9b15e33df4e342d5d8e9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-25 13:38:04.000000000","message":"Patch Set 28:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-28/ns5_nfs : SUCCESS in 3h 30m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-28/ns5_iscsi : SUCCESS in 3h 31m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":28},{"id":"00fc8a676a966df5429181519e175001a2204ae9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-25 13:40:16.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/28/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/6565f7d/ : SUCCESS in 52m 42s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":28},{"id":"19500eabfb4bc39c6a9c619493641afbbfdbe576","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-25 13:43:55.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/28/upstream-check/cinder-tempest-plugin-ontap-fcp/2ee0006/ : SUCCESS in 56m 21s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/28/upstream-check/cinder-tempest-plugin-ontap-iscsi/e60cf53/ : SUCCESS in 42m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/28/upstream-check/cinder-tempest-plugin-ontap-nfs/8df673d/ : FAILURE in 40m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":28},{"id":"480a26f153551b26ddb7740f6cf24ab05819e1e5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-25 20:30:13.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/28/ : SUCCESS in 1h 41m 48s","accounts_in_message":[],"_revision_number":28},{"id":"7cf5aa2343ad42e55c58a065ac5d6fac974c5c6e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-27 10:55:18.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/28/20221127161857/NEC-ISCSI : SUCCESS in 3h 35m 01s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/28/20221127161857/NEC-FC : SUCCESS in 3h 36m 18s","accounts_in_message":[],"_revision_number":28},{"id":"94f899b535790e48ec32cbf5e1f0eecc79f7c6e7","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-28 15:08:11.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"4bdac74788c406e16fb50f4cbded20cbe0d76a6b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-29 03:24:24.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/28/20221129061227/NEC-V-ISCSI : SUCCESS in 2h 55m 44s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/28/20221129061227/NEC-V-FC : SUCCESS in 6h 11m 53s","accounts_in_message":[],"_revision_number":28},{"id":"872abf05bbf259ebb93d4d12d9244c07be14da7e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-11-29 22:58:45.000000000","message":"Patch Set 28: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"e9408db48493f731d1d12b7e9f0c8d086ecde46d","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-12-01 13:09:32.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"66c422b587e8955c447281f60e468b9c9cfec702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-12-01 13:11:30.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"04a5ab85a341f568d27759a8c9c1d26bdfa06079","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-01 13:11:38.000000000","message":"Patch Set 29:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1817 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"1a4b3bc59bca8be3474c86588fc5a052546f324e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-12-01 13:29:10.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":29},{"id":"c011712b7ee29a1460829319b4a5513e353634f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 13:41:45.000000000","message":"Patch Set 29:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/835512 is needed.","accounts_in_message":[],"_revision_number":29},{"id":"1b36d1128824698a8e4a65aa8a652604da5d5d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 13:41:46.000000000","message":"Patch Set 29: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/835512 is needed.","accounts_in_message":[],"_revision_number":29},{"id":"fcbb1870697b097abfcf98c1b61bcf02307ef5bf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-01 13:41:47.000000000","message":"Patch Set 29:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"663d79697367f438eecb516ce82e206fa1357ccc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-01 13:41:47.000000000","message":"Patch Set 29:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":29},{"id":"0ba694d7906fdbb9c61393c03f7cddd22064fac4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-12-01 13:41:53.000000000","message":"Patch Set 29:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"93c134898658bc8cd9a986ebe71713b7705edfa2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-01 13:41:58.000000000","message":"Patch Set 29:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"2abd255bf3c9f58ecd8ea016a292fea780719541","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-01 13:42:00.000000000","message":"Patch Set 29:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"eef6e417c0dc9e19bed16cdceaaa76f23d8bcbee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-01 13:42:01.000000000","message":"Patch Set 29:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"ea2803360c7677add8b908f908e629ed1e54aa87","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-01 13:42:04.000000000","message":"Patch Set 29:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"f67a1d5ab72b6a997cf5fc460739aba003232dd4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-12-01 13:54:43.000000000","message":"Patch Set 29: Code-Review+1","accounts_in_message":[],"_revision_number":29},{"id":"c549460a4d759f616930726154e179ae8916c878","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-01 16:42:26.000000000","message":"Patch Set 29:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-29/ns5_nfs : SUCCESS in 3h 30m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-29/ns5_iscsi : FAILURE in 26m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":29},{"id":"6ad80f8a8ea500d7e424d8f30afea5a3c4c777a1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-12-01 16:48:02.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/29/20221201221133/NEC-FC : SUCCESS in 3h 35m 31s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/29/20221201221133/NEC-ISCSI : SUCCESS in 3h 36m 26s","accounts_in_message":[],"_revision_number":29},{"id":"94c3e34feec49f2babe418cd96aacc44312f1e16","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-02 19:50:09.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/29/20221202214846/NEC-V-FC : SUCCESS in 5h 47m 06s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/29/20221202214846/NEC-V-ISCSI : SUCCESS in 7h 01m 20s","accounts_in_message":[],"_revision_number":29},{"id":"b6ef1f8e0129e3ac766848b6251c4dd2451dd20f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-01-23 14:00:52.000000000","message":"Patch Set 30: Commit message was updated.","accounts_in_message":[],"_revision_number":30},{"id":"384d6f8e638a5140d8ede13fe69540e5f7fef29c","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-23 14:01:38.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":30},{"id":"dd6dd233acbde29beddd30db5882667fc935430c","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-23 14:16:16.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/71e9188983154d6088d7daded72f8250 : POST_FAILURE in 14m 37s","accounts_in_message":[],"_revision_number":30},{"id":"dd89aba5b7800080cd8d76489defc030bf232e23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 14:33:52.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f212bdd71914403dbad01286f8279dab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee125ca19312419aa2b439d14d853646 : SUCCESS in 29m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dea818cf0c144000bb909706fa053293 : FAILURE in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"187d28e0cbc86564f3d5d23202157fa96fb6666a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-23 15:08:38.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/30/ : FAILURE in 59m 39s","accounts_in_message":[],"_revision_number":30},{"id":"bcf58811df0be09e178e470cc5784bbc258986fc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-23 15:15:25.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/30/check/iscsi-driver-alletra-6k/6d3991f/6d3991f.html : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":30},{"id":"041bd43ca0848b13c1a116eff651f390a4cdfdec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-23 15:22:57.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b1254d5c43a94978b20e1b582c931c7c : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":30},{"id":"7688529f9ed936bd9d82f638eb131966172c3be7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-23 15:26:32.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : FAILURE in 23m 33s","accounts_in_message":[],"_revision_number":30},{"id":"2cf4623b5b16944ce80c6bbb22d842c041659d9c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-23 16:07:17.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : FAILURE in 26m 12s","accounts_in_message":[],"_revision_number":30},{"id":"ddef6b329150a75093503a5b260b3acc2c029370","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-23 16:07:37.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/30/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/920de01/ : SUCCESS in 1h 48m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/30/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6aed4a7/ : FAILURE in 2h 05m 04s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/750782/30/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/57f1560/ : FAILURE in 2h 05m 04s","accounts_in_message":[],"_revision_number":30},{"id":"1a7b5dccdfc6632073709c10ac8ba2306699c5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:36:48.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e647149eb24af198bf1faebb570e54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91e87285e4744bbfab94b6cb41235177 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e09bee83921e4b00a4df0820bdf7af7e : SUCCESS in 12m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bed0108ea6184f17a8fbdab4a49b2ac9 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c13abae7224706b7b37f7f66d1e6bb : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/bc7e72ec1631472cbb15a70a0b680642 : SUCCESS in 39m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/00e1d4ee980940b3b5e7498967658b52 : SUCCESS in 55m 49s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2e8513a1691148318b80d140feb10cc1 : SUCCESS in 1h 19m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff1aa28563d44623aaedebfe3bbfc6cb : SUCCESS in 44m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9004087565304685a142334952c89a66 : SUCCESS in 7m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9e83c795f48b4944bec54df049c74748 : SUCCESS in 17m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/67bf2a8ab35a4876b11aef4cc5b47d3e : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8b68a2ca72f43a28bdda0bb9c479c49 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ef5e82f861274698a280f306544b7078 : SUCCESS in 5m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a9e10c52f64644f685d738089521a185 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2cb6f8118c1b4c808c2a79c66e581e95 : SUCCESS in 53m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e33ba57a1e064a57b76a56565e148491 : FAILURE in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a7a23096d404baa8ae0c7dc829f4716 : SUCCESS in 1h 14m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/680a3004a4cc462f9caabbe1e3ec459f : FAILURE in 1h 19m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13e7a11a1a8041bcbe5059e7b83b1ee9 : SUCCESS in 1h 54m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2571e2cfd23b487e8d280ca73cf79de1 : SUCCESS in 1h 09m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09da9c04b02646318c591d295d48eae5 : SUCCESS in 1h 42m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dfdc741fce66480c989119bf0e5a1eaf : SUCCESS in 40m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/296da7549efe48d7b6f3a7d28ef00065 : SUCCESS in 1h 29m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8b2234f707143a9b5b5b0190fdef622 : SUCCESS in 2h 22m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bafa37585984d4cad93c02ede9b4d25 : SUCCESS in 56m 47s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5bbb00b45e1418fb64b00fd1b909c6d : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":30},{"id":"f46fc217a69f17f239202338361fd641236a9a19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-23 16:44:34.000000000","message":"Patch Set 30:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_01_2023_16_37_00\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":30},{"id":"541aa92bad9ddd1bb8dd121e3b667266a3cdfc64","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-23 17:18:15.000000000","message":"Patch Set 30:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c4c1428e8a55480e84bfe40f7511fbfc : POST_FAILURE in 1h 32m 53s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/9454bdc7c34241e7ad96285cdda45ca7 : POST_FAILURE in 1h 18m 04s","accounts_in_message":[],"_revision_number":30},{"id":"56ee5736462f33e8b80ec53689bf69d57d3fe2fb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-23 17:41:00.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/30/upstream-check/cinder-tempest-plugin-ontap-fcp/ae5934d/ : SUCCESS in 1h 02m 02s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/30/upstream-check/cinder-tempest-plugin-ontap-iscsi/b634c57/ : SUCCESS in 43m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/30/upstream-check/cinder-tempest-plugin-ontap-nfs/feacc28/ : SUCCESS in 44m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/750782/30/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/38f9fa7/ : FAILURE in 31m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":30},{"id":"79e911e9cc8e623f65a740801f6d628bdf1bbc3e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-23 21:02:52.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/750782-30/HPMSA-iSCSI/ : SUCCESS in 1:38:14\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/750782-30/HPMSA-FC/ : SUCCESS in 1:17:31","accounts_in_message":[],"_revision_number":30},{"id":"4db393dd4941cc9d271fa9be10ab96c2f5d1e0f3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-23 22:00:12.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/750782-30/Seagate-iSCSI/ : SUCCESS in 1:54:02\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/750782-30/Seagate-FC/ : SUCCESS in 1:32:19","accounts_in_message":[],"_revision_number":30},{"id":"f3322bff0ca7e6c0cb32d38b4674545bfbc60f21","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-23 22:07:25.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/750782-30/PVME-iSCSI/ : SUCCESS in 1:52:20\n- PVME-FC https://os-logs.seagate.com/dell-ci/750782-30/PVME-FC/ : SUCCESS in 1:34:58","accounts_in_message":[],"_revision_number":30},{"id":"a4c04b9b33f6d30694a0b734671e29e379ba1dda","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-24 03:33:11.000000000","message":"Patch Set 30:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-30/ns5_nfs : FAILURE in 12h 20m 35s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-30/ns5_iscsi : FAILURE in 13h 31m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":30},{"id":"8f01ab07bc21a2d5d7b7225441faf19eebc10e92","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-09 19:13:32.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/30/20230210034439/NEC-V-FC : SUCCESS in 28m 05s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/30/20230210034439/NEC-V-ISCSI : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":30},{"id":"a6c6fcfaba32fe2f0da77be18f51f70290486e7f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-15 03:31:46.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/30/20230215085654/NEC-ISCSI : SUCCESS in 3h 34m 13s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/30/20230215085654/NEC-FC : SUCCESS in 3h 34m 48s","accounts_in_message":[],"_revision_number":30},{"id":"2788c38590fb61635a5a4d7113a952fb0d221e0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-02-20 14:55:04.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"1f3b1b8abfd2f6ac8c1a375b8a5d4016ba5bdcc9","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-02-20 15:04:40.000000000","message":"Patch Set 31: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"3a5ec981339e388a43e701ee0e07d23e47ef024e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:06:26.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc94b07250f4002bd2587232d170f88\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff8658b2f9d04f469e7651a89dfda0eb : FAILURE in 10m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f18af34bd3b4d198d15534e43f8995b : FAILURE in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"5709465d24fe8210bb2b713a6c52e78de6fe7c01","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-20 15:21:09.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : FAILURE in 23m 50s","accounts_in_message":[],"_revision_number":31},{"id":"e67ec3e0e147fad672c5e89a464400da139c913b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-20 15:23:51.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/31/check/iscsi-driver-alletra-6k/9c77de5/9c77de5.html : FAILURE in 26m 12s","accounts_in_message":[],"_revision_number":31},{"id":"6605356e659c1ba436423076eed3a8f87f00d3fc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-20 15:23:56.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : FAILURE in 26m 49s","accounts_in_message":[],"_revision_number":31},{"id":"cebae0c5c257d5065de67b92ed97e4776762092d","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-20 16:02:55.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/5486553af8f74eefae2bbbe341b9e445 : POST_FAILURE in 1h 07m 05s","accounts_in_message":[],"_revision_number":31},{"id":"baa98996b304a50f431edcddd57bafd7fa285596","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-20 16:05:15.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/31/ : FAILURE in 1h 02m 18s","accounts_in_message":[],"_revision_number":31},{"id":"4e1a4557eb04e28984680693ae6ffbd2c4bf7018","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-20 16:31:15.000000000","message":"Patch Set 31:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/82/750782/31/check/fujitsu-eternus-dx-iscsi/f0fbd3a : FAILURE in 46m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/82/750782/31/check/fujitsu-eternus-dx-fc/366d51d : SUCCESS in 1h 35m 29s","accounts_in_message":[],"_revision_number":31},{"id":"6f2531d9581a38746c6a707826eac5c1cefddefa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-20 16:54:27.000000000","message":"Patch Set 31:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/31/check/DellEMC_Unity_FC/63ae566/DellEMC_Unity_FC/2327 : SUCCESS in 1h 20m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/31/check/DellEMC_Unity_iSCSI/a87eca5/DellEMC_Unity_iSCSI/2181 : SUCCESS in 1h 36m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":31},{"id":"a49ae177e6955c8a874e63c86b7ac6056ce9dd03","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-20 17:01:28.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/31/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a80e692/ : SUCCESS in 1h 55m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/31/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0738af9/ : FAILURE in 2h 03m 23s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/82/750782/31/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/f194f28/ : SUCCESS in 2h 04m 22s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/82/750782/31/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/9d2364e/ : SUCCESS in 1h 54m 18s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"9374d7366c102bbf1e47cf0eefc05e49211dbc24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 18:31:54.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95b6fb8676154a92a437c6538d393aac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189c5540a6a24ba3ac48be782929ecbe : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34336d9f98c14c3887c8fa56e5fc1104 : FAILURE in 13m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68feb24a126e4774911b83d082f32502 : FAILURE in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf5163feb604c08bcf6a66d3502393f : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/fc665fb41d3b48d29fbc9aa6ba7d7419 : SUCCESS in 1h 05m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/94446f1b2be84dd7af66243ba1ca4cf4 : SUCCESS in 1h 07m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57bbbf82f5cb4735b78ffcade3a5e295 : SUCCESS in 1h 42m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8285ae66ef76496fbdc1822f0d73661c : SUCCESS in 50m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c62a4716c2f4411bb13f1a8ca911c82 : SUCCESS in 8m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6657eef2e51c4d58a0d2ecfd10146f5f : FAILURE in 13m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a588869b1be243e68a5c2c1e49333ca5 : FAILURE in 6m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/64abc962466d4eeaa17cd4eb55a8e1ca : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e0b51226d43a4df3a0fa829aedd30e1e : SUCCESS in 5m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57adb8f002fc40858595656ee78f8382 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1d515504fcf54f7dbdf95f5bedcba61a : SUCCESS in 47m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c4306ee412a64784aa8b0d7f477cb726 : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2f4ace9f24b64e87b1aa1b7bc251e4df : SUCCESS in 1h 24m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6674e0b03b1c4d7885f0ba6bffa68da3 : FAILURE in 1h 22m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/733132e4bd1b4a1dab08e224f64a19ec : SUCCESS in 1h 24m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c798e137a814f0fac2d76da861ebe16 : SUCCESS in 1h 05m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5f0dd72b4d0d44508a58686503798fae : SUCCESS in 1h 38m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/29ea0f144553429bbbf1145444e6afca : SUCCESS in 46m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bd89aad90fd940aa80d88b51f7b046d5 : SUCCESS in 1h 22m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c4d5b4fa1e1f49c1bfdc8321f514b1b5 : SUCCESS in 2h 33m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76640c8007ee4678a7faaf80a1e71683 : SUCCESS in 54m 29s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6adae7f080e490482d667470b186522 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":31},{"id":"7d7ab46c1668cda4cc0d850dc2129e848c273811","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-20 22:27:36.000000000","message":"Patch Set 31:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-31/ns5_nfs : FAILURE in 33m 15s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-31/ns5_iscsi : FAILURE in 7h 32m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":31},{"id":"fb28904a46dbcc8b6005f87c4c8bbd1b153bc356","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-20 23:15:27.000000000","message":"Patch Set 31:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/6936ea69739842048689be925f1230b1 : SUCCESS in 1h 50m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/c7e03a8d16184908ad5b3df6e99731c3 : FAILURE in 1h 37m 33s","accounts_in_message":[],"_revision_number":31},{"id":"bf36d3b9f0ca521c7ad4fb8d188dc5e91b5c8e9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-02-21 12:08:16.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"8a51cb2622add73255d2e5b63ec7301d6d6da99b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-02-21 12:09:47.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":32},{"id":"9b81290a39b0bc8dc65b86496a7501b639e5245f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 12:18:16.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4212f7057884d4c8063f6dc7fbc65de\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d4d364e9a694c5699116836fd3a4a9f : FAILURE in 8m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/92d200539777451bae5168617718c8cf : FAILURE in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"9d8ab000a0c07a5b7d1310e0c14942c5a226349a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-21 12:39:23.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/32/check/iscsi-driver-alletra-6k/39db75f/39db75f.html : FAILURE in 25m 21s","accounts_in_message":[],"_revision_number":32},{"id":"221645fe9aa99b43dc36be3554576a40ab77416c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-21 12:41:11.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/750782 : FAILURE in 23m 31s","accounts_in_message":[],"_revision_number":32},{"id":"b45905e286a9d4f9375ad58aef5bdb8c40faf146","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-21 12:44:14.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : FAILURE in 26m 04s","accounts_in_message":[],"_revision_number":32},{"id":"1d0e9b8dc9309d712b95bbf2ff787cb74786229b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-21 12:55:25.000000000","message":"Patch Set 32:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/82/750782/32/check/fujitsu-eternus-dx-iscsi/314fd26 : FAILURE in 46m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/82/750782/32/check/fujitsu-eternus-dx-fc/c40dabb : FAILURE in 3m 12s","accounts_in_message":[],"_revision_number":32},{"id":"01b835b0e713006d658003c604271b03af9c203a","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-21 14:31:25.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/db76257ed2e042639b45025169a8b88a : POST_FAILURE in 33m 11s","accounts_in_message":[],"_revision_number":32},{"id":"716eee956617eaa840caa82ae13c763590b30976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 14:49:24.000000000","message":"Patch Set 32: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d41340280494ef4ad0dbeb7fea38d40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09c3447b1b154c60993335b9bbf7d092 : FAILURE in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bdfdf4ed7ea4656b3d0e2229a68f6b5 : FAILURE in 11m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8270b82ad654b6a81fc767bbf7aed96 : FAILURE in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece3cdab3a0a4421916b629654c9c117 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/7027c9f13a9741e982353cffa12090f6 : SUCCESS in 52m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e6aa7b3c860f449b87c308ec2214a921 : SUCCESS in 57m 34s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab64b60cf2d94884a79808f8f60e9fa9 : SUCCESS in 1h 24m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25c3045585ed4142bc4a5487aac465f8 : SUCCESS in 57m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac8d8ddb09de4bc991487d49e2486b19 : SUCCESS in 8m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3210dd0e25d448b1bfb20109d6c22a95 : FAILURE in 15m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5d752b9180d74331a3dc09c9939731ab : FAILURE in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5b5ffe8e5f145c69a326ab26d706947 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/94d42f680f354a9cb8dfb615e2b76074 : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12d2c94c0bf447ea9a8ae6f01974ae87 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a03d458ce17540b99be05f0d0bc6e752 : SUCCESS in 35m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3669aa811b874a7da2d4a0efe0988928 : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc72cfbf24fa4ebda3c7e2c52a5e83b0 : SUCCESS in 1h 14m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/42d1230bfec24c869fb2576991305ac8 : FAILURE in 2h 10m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/74b9f0457f644d0a878910482d3fe0d1 : FAILURE in 1h 30m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/06cdde994758419489c6c2a24c00b046 : SUCCESS in 1h 08m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/06e7e28a1a81444880bc88c212efb3a7 : SUCCESS in 1h 37m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c28e37b7acac4d0fabd9547cd760fd74 : SUCCESS in 35m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/36d60ff07c104a0b8790fb3f6955323b : SUCCESS in 1h 12m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ddf655dae184fdc986ba24b7d9f3a69 : FAILURE in 2h 26m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9971cd78835a44f891362c4d967f4fa1 : SUCCESS in 1h 11m 36s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43298cf499f844e38df380de975d8abd : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":32},{"id":"9c7ad59889f82915cd2824cca7f96679440a30a8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-21 15:07:59.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/750782/32/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/df9cb6d/ : SUCCESS in 1h 44m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/750782/32/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d984536/ : FAILURE in 2h 57m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/82/750782/32/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/7363a35/ : SUCCESS in 2h 41m 11s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/82/750782/32/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/d35faff/ : SUCCESS in 1h 57m 30s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"d6e756d4beb7636566dcd6a2d78b675e37454817","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-21 16:06:05.000000000","message":"Patch Set 32:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2905d7afc0d742d49a93f0a0554d1d38 : FAILURE in 2h 08m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ad0138d4d361495c9f0b3d89c2dc634b : FAILURE in 1h 37m 46s","accounts_in_message":[],"_revision_number":32},{"id":"7b988da893dff89a4424e08e33460c590285fb1f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-21 16:08:19.000000000","message":"Patch Set 32:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/32/check/DellEMC_Unity_FC/be60323/DellEMC_Unity_FC/2331 : SUCCESS in 1h 30m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/32/check/DellEMC_Unity_iSCSI/eaeef64/DellEMC_Unity_iSCSI/2185 : SUCCESS in 2h 01m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":32},{"id":"ad72bd7d5e922c0d2e8516f8ea2a12b962d4d52f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-21 17:45:48.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/32/ : FAILURE in 56m 42s","accounts_in_message":[],"_revision_number":32},{"id":"a97e43408c80c92222726b975d2831493a8e383e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-21 22:48:26.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4695ae901fa4428da9633a500af7ee53 : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":32},{"id":"c1a363180fae8edc62c92cf8bab830610a1033d7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-22 03:44:18.000000000","message":"Patch Set 32:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-32/ns5_nfs : FAILURE in 14h 41m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-32/ns5_iscsi : FAILURE in 40m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":32},{"id":"7cc754365d436d68225277d177af1039a4018f92","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-24 04:34:49.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/32/20230224105909/NEC-V-FC : SUCCESS in 2h 22m 49s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/32/20230224105909/NEC-V-ISCSI : SUCCESS in 2h 35m 36s","accounts_in_message":[],"_revision_number":32},{"id":"7af8abcb09c455394dda7b33b47d980f99ff290b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-01 06:52:22.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/32/20230301122137/NEC-ISCSI : SUCCESS in 3h 29m 57s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/32/20230301122137/NEC-FC : SUCCESS in 3h 30m 42s","accounts_in_message":[],"_revision_number":32},{"id":"076ed82b0709debc3ce32a2d98817d9294b839d4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 08:09:31.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":32},{"id":"6545f951481276462c97fcbe296c2dd8ccce6762","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-04-20 12:58:53.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"72c3c01caaaabd4d13a06530f565ca15141bab72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 13:15:17.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b88a86f5acb8465b82085003f666a5ee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b49f7c209eb4e9ca998be16d2b92749 : FAILURE in 10m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/237783c2a38e4269a5a0ed927e82b926 : FAILURE in 9m 12s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"5ef7fd24b2a03fa905b0fe0c065eadf5e67a2f6f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-20 14:04:13.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a403ca3439284502b1b8b1ec870d01de : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":33},{"id":"b02b36005539d767d7a59eba194fe906070161d9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-20 14:27:00.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/750782 : SUCCESS in 1h 25m 51s","accounts_in_message":[],"_revision_number":33},{"id":"157e1d927f5ad4e6f639f1238f38b17624af8751","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-20 14:39:03.000000000","message":"Patch Set 33:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/6514247059c04a8bb9ca877239de9489 : POST_FAILURE in 46m 45s","accounts_in_message":[],"_revision_number":33},{"id":"424bce6904feafd17e065893bfe13d65edc2b1f1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-20 14:47:01.000000000","message":"Patch Set 33:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/750782/33/check/iscsi-driver-alletra-6k/47b4e65/47b4e65.html : FAILURE in 1h 45m 59s","accounts_in_message":[],"_revision_number":33},{"id":"e0b33cd53c73a065d936f8ab74b469cbef37ff2d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-20 14:54:35.000000000","message":"Patch Set 33:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b58a94e42e2e434eacbdeca68aeb9566\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/50692b74adcc479bbaf619ce797a182c : POST_FAILURE in 1h 00m 03s","accounts_in_message":[],"_revision_number":33},{"id":"5179a610e0d6e1a3e2f6bf922c99692303a0ceb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 15:12:36.000000000","message":"Patch Set 33: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b86c9f5bdd334283b7a6ab854e5ae159\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1063e229cb374244bd1ad089d5a7f51e : FAILURE in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f087a0e50924aafabfd04de858d6809 : FAILURE in 12m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44569c4b98454fd88237b19560907746 : FAILURE in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b25f4fa0294c90974b64463ad3528c : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/7659e4eb631f4cbca016637a6aebb4de : SUCCESS in 1h 16m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b032821419394e7ea3b3894d103c011e : SUCCESS in 1h 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3fddfd18aed476eb42b8c9d8e512e9b : SUCCESS in 57m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa55453b05af45819b66f18aab857a35 : SUCCESS in 8m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/75435b22e00f47f2b47937c406202e55 : FAILURE in 13m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/af4b02d2b9684ffeab8b9b97e3ac3095 : FAILURE in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/102fac9475a3444384612d16a8013762 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5d0bd0e38e6b438b9883d5a4182b2b5d : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea4183a2fcf34176a8406dffd71109d4 : SUCCESS in 5m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/589f414773494afe90ce3a63f391fbe1 : SUCCESS in 52m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e3aabd0e5a04da0b42b925387eb72a1 : SUCCESS in 8m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eaf5ed9e25e04dc3ae8ebb10273c313c : SUCCESS in 1h 32m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2d4a4cb6f06447d992a7873c780bfcfd : FAILURE in 1h 17m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f5957073bd034493a2bae518ade783fc : SUCCESS in 1h 42m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/44a114ec6ba24f668bcb7e7541b5f403 : SUCCESS in 2h 03m 25s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f4c348ddf6e7458a9efa2f774744905e : SUCCESS in 59m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9e7881732d744680b1fef2a14bd4efec : SUCCESS in 1h 24m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b75e0d486509423482a209f0692738a7 : SUCCESS in 44m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0ab0d36b78cc49259c67de97b0706016 : SUCCESS in 1h 22m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bd44369a0c7447bc8d1c47277de7c155 : SUCCESS in 1h 15m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4766fdd2ad0144d08a10176b6fc65d29 : SUCCESS in 2h 11m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75a4944da7704b31a5f47bffc6c8484d : SUCCESS in 50m 14s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/904be511f89e433cb7f507f2733a1cc8 : SUCCESS in 1h 27m 17s","accounts_in_message":[],"_revision_number":33},{"id":"81ee50f7413c565fa687cba7bea69f7f20f35626","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-20 15:27:43.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/82/750782/33/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/a36f12a/ : SUCCESS in 1h 59m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/82/750782/33/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/7282b1d/ : FAILURE in 2h 26m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/82/750782/33/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/5a45bbc/ : SUCCESS in 2h 00m 09s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/82/750782/33/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/0af9624/ : FAILURE in 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"31a21ebf0f88a779b18e751e982af6f0090612dd","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-20 15:32:42.000000000","message":"Patch Set 33:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/750782/33/ : FAILURE in 46m 44s","accounts_in_message":[],"_revision_number":33},{"id":"b349c0ea68d7a51fac1866e6960f2fe268b803ac","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-20 16:20:41.000000000","message":"Patch Set 33:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-82-750782-33/FC/ : SUCCESS in 3h 13m 49s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-82-750782-33/iSCSI/ : SUCCESS in 3h 8m 12s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":33},{"id":"5c0705148038a0c470a883995957f7eef7d3d197","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-20 16:43:35.000000000","message":"Patch Set 33:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-750782-33/ns5_nfs : SUCCESS in 3h 24m 21s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-750782-33/ns5_iscsi : SUCCESS in 3h 29m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":33},{"id":"7a7b1e5f0c54dcc8b56c54b2a0ee9a239afdb8fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-20 17:06:52.000000000","message":"Patch Set 33:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/750782/33/check/DellEMC_Unity_FC/a2979ab/DellEMC_Unity_FC/48 : SUCCESS in 1h 29m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/750782/33/check/DellEMC_Unity_iSCSI/01be246/DellEMC_Unity_iSCSI/50 : SUCCESS in 2h 12m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":33},{"id":"a5c9c1b844bee44a005f27fa3befc85485febc18","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-21 22:21:09.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/33/20230422044445/NEC-V-ISCSI : SUCCESS in 2h 29m 24s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/33/20230422044445/NEC-V-FC : SUCCESS in 2h 36m 21s","accounts_in_message":[],"_revision_number":33},{"id":"cf0e0f2626b5e3d5307b3aa56547b0b50cd1b194","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-22 08:44:19.000000000","message":"Patch Set 33:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9e4779f0c13c43a2a351b7cd47ec9fd3 : RETRY_LIMIT in 1m 04s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/82/750782/33/check/devstack-plugin-ceph-dsvm-focal-py3/024abc0/ : SUCCESS in 2h 02m 04s","accounts_in_message":[],"_revision_number":33},{"id":"4cf92f960f18a8fd711e33d820e6018106f817fc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-22 19:27:56.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/33/20230423005642/NEC-ISCSI : SUCCESS in 3h 24m 17s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/750782/33/20230423005642/NEC-FC : SUCCESS in 3h 31m 11s","accounts_in_message":[],"_revision_number":33},{"id":"1c4fe0d6ed4e34ec95e272cc5da8c2da049f9760","tag":"autogenerated:gerrit:abandon","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-04-24 16:43:16.000000000","message":"Abandoned\n\nI have decided not to continue with the refactor of the Ceph backup driver for this bug. Instead, I am currently working on the bug mentioned in https://review.opendev.org/c/openstack/cinder/+/880965, and I plan to follow up with a refactor/cleaning of the code for this driver in another patch.","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"4e98f5447fb65f4de2dafbc0bad09d875acb578d","revisions":{"d90deb03022d49da0bb02bdf993a4a042a0a3155":{"kind":"REWORK","_number":1,"created":"2020-09-09 18:31:02.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/1"}}},"commit":{"parents":[{"commit":"06bc733ffae14782b8fcbc509c07ee6822b74122","subject":"Merge \"Add release note for zstd compression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06bc733ffae14782b8fcbc509c07ee6822b74122"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:30:39.000000000","tz":0},"subject":"Add option for no rbd volumes in the ceph backup driver","message":"Add option for no rbd volumes in the ceph backup driver\n\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\nCloses-Bug: #1895035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d90deb03022d49da0bb02bdf993a4a042a0a3155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d90deb03022d49da0bb02bdf993a4a042a0a3155"}]},"branch":"refs/heads/master"},"b58e0a16a51cbbf925bdc7162f7ec659e6e4f563":{"kind":"REWORK","_number":2,"created":"2020-10-05 21:51:39.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/2"}}},"commit":{"parents":[{"commit":"06bc733ffae14782b8fcbc509c07ee6822b74122","subject":"Merge \"Add release note for zstd compression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06bc733ffae14782b8fcbc509c07ee6822b74122"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-10-05 21:51:27.000000000","tz":0},"subject":"Add option for no rbd volumes in the ceph backup driver","message":"Add option for no rbd volumes in the ceph backup driver\n\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\nCloses-Bug: #1895035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b58e0a16a51cbbf925bdc7162f7ec659e6e4f563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b58e0a16a51cbbf925bdc7162f7ec659e6e4f563"}]},"branch":"refs/heads/master"},"ab6cd2a0a47467275eaa1eb24d7a3fe15c5c949e":{"kind":"REWORK","_number":3,"created":"2020-11-19 19:57:09.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/3"}}},"commit":{"parents":[{"commit":"06bc733ffae14782b8fcbc509c07ee6822b74122","subject":"Merge \"Add release note for zstd compression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06bc733ffae14782b8fcbc509c07ee6822b74122"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-11-19 19:55:34.000000000","tz":0},"subject":"Add option for no rbd volumes in the ceph backup driver","message":"Add option for no rbd volumes in the ceph backup driver\n\nCurrently the Ceph backend  fails when restoring non-ceph\nvolumes. This happens because the current code is\ngenerating the wrong base image when performing a full\nrestore.  If we are not using a RBD destination volume we\nneed to specify the backup in the same way than If a source\nsnapshot is provided.\n\nIn order to locale the right base image used for the backup\nwe should use the one specified in the backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab6cd2a0a47467275eaa1eb24d7a3fe15c5c949e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab6cd2a0a47467275eaa1eb24d7a3fe15c5c949e"}]},"branch":"refs/heads/master"},"25ea0c72afec77a89c9149329f62d829c7f60c02":{"kind":"REWORK","_number":4,"created":"2020-11-20 02:37:56.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/4"}}},"commit":{"parents":[{"commit":"06bc733ffae14782b8fcbc509c07ee6822b74122","subject":"Merge \"Add release note for zstd compression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06bc733ffae14782b8fcbc509c07ee6822b74122"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-11-20 02:37:50.000000000","tz":0},"subject":"Add option for no rbd volumes in the ceph backup driver","message":"Add option for no rbd volumes in the ceph backup driver\n\nCurrently the Ceph backend  fails when restoring non-ceph\nvolumes. This happens because the current code is\ngenerating the wrong base image when performing a full\nrestore.  If we are not using a RBD destination volume we\nneed to specify the backup in the same way than If a source\nsnapshot is provided.\n\nIn order to locale the right base image used for the backup\nwe should use the one specified in the backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25ea0c72afec77a89c9149329f62d829c7f60c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25ea0c72afec77a89c9149329f62d829c7f60c02"}]},"branch":"refs/heads/master"},"ad04d99601fa09509b17efb779fb53216d903477":{"kind":"REWORK","_number":5,"created":"2021-01-19 21:32:16.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/5"}}},"commit":{"parents":[{"commit":"06bc733ffae14782b8fcbc509c07ee6822b74122","subject":"Merge \"Add release note for zstd compression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06bc733ffae14782b8fcbc509c07ee6822b74122"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-01-19 21:32:07.000000000","tz":0},"subject":"Fix restore Ceph backups to no-RBD volumes","message":"Fix restore Ceph backups to no-RBD volumes\n\nCurrently the Ceph backend fails when restoring no-ceph volumes.\nThis happens because the current code is generating the wrong base\nimage when performing a full restore.\nIf we are not using a RBD destination volume we need to specify the\nbackup in the same way than if a source snapshot is provided.\nIn this way _get_backup_base_name returns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad04d99601fa09509b17efb779fb53216d903477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad04d99601fa09509b17efb779fb53216d903477"}]},"branch":"refs/heads/master"},"b72b7c4a2b393db76180b51c2a25eac86bdc25b5":{"kind":"REWORK","_number":6,"created":"2021-09-02 16:40:16.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/6"}}},"commit":{"parents":[{"commit":"06bc733ffae14782b8fcbc509c07ee6822b74122","subject":"Merge \"Add release note for zstd compression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06bc733ffae14782b8fcbc509c07ee6822b74122"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-09-02 16:39:33.000000000","tz":0},"subject":"Fix restore Ceph backups to no-RBD volumes","message":"Fix restore Ceph backups to no-RBD volumes\n\nCurrently the Ceph backend fails when restoring no-ceph volumes.\nThis happens because the current code is generating the wrong base\nimage when performing a full restore.\nIf we are not using a RBD destination volume we need to specify the\nbackup in the same way than if a source snapshot is provided.\nIn this way _get_backup_base_name returns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b72b7c4a2b393db76180b51c2a25eac86bdc25b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b72b7c4a2b393db76180b51c2a25eac86bdc25b5"}]},"branch":"refs/heads/master"},"6790147031e90b6bf247803e87eead031501e748":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-09-02 16:40:32.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/7"}}},"commit":{"parents":[{"commit":"6258cc99f95f387d1ff9f6922d48e1651b2d2e90","subject":"Merge \"Snapshot in-use volumes without force flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6258cc99f95f387d1ff9f6922d48e1651b2d2e90"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-09-02 16:40:32.000000000","tz":0},"subject":"Fix restore Ceph backups to no-RBD volumes","message":"Fix restore Ceph backups to no-RBD volumes\n\nCurrently the Ceph backend fails when restoring no-ceph volumes.\nThis happens because the current code is generating the wrong base\nimage when performing a full restore.\nIf we are not using a RBD destination volume we need to specify the\nbackup in the same way than if a source snapshot is provided.\nIn this way _get_backup_base_name returns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6790147031e90b6bf247803e87eead031501e748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6790147031e90b6bf247803e87eead031501e748"}]},"branch":"refs/heads/master","description":"Rebase"},"37956382d276bbfa8c8fe386a1e4ed6cc852c90e":{"kind":"REWORK","_number":8,"created":"2021-09-02 17:33:11.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/8"}}},"commit":{"parents":[{"commit":"6258cc99f95f387d1ff9f6922d48e1651b2d2e90","subject":"Merge \"Snapshot in-use volumes without force flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6258cc99f95f387d1ff9f6922d48e1651b2d2e90"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-09-02 17:31:51.000000000","tz":0},"subject":"Fix restore Ceph backups to no-RBD volumes","message":"Fix restore Ceph backups to no-RBD volumes\n\nCurrently the Ceph backend fails when restoring no-ceph volumes.\nThis happens because the current code is generating the wrong base\nimage when performing a full restore.\nIf we are not using a RBD destination volume we need to specify the\nbackup in the same way than if a source snapshot is provided.\nIn this way _get_backup_base_name returns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37956382d276bbfa8c8fe386a1e4ed6cc852c90e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37956382d276bbfa8c8fe386a1e4ed6cc852c90e"}]},"branch":"refs/heads/master"},"61617a3d3e0a6a3a951eb41a1201924ee0724cbc":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-11-12 20:21:53.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/9"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-11-12 20:21:53.000000000","tz":0},"subject":"Fix restore Ceph backups to no-RBD volumes","message":"Fix restore Ceph backups to no-RBD volumes\n\nCurrently the Ceph backend fails when restoring no-ceph volumes.\nThis happens because the current code is generating the wrong base\nimage when performing a full restore.\nIf we are not using a RBD destination volume we need to specify the\nbackup in the same way than if a source snapshot is provided.\nIn this way _get_backup_base_name returns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/61617a3d3e0a6a3a951eb41a1201924ee0724cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/61617a3d3e0a6a3a951eb41a1201924ee0724cbc"}]},"branch":"refs/heads/master","description":"Rebase"},"db73d7f55b09146c48fde85767121f6bd3c14c9b":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-01-10 20:22:20.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/10"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-01-10 20:22:20.000000000","tz":0},"subject":"rbd: fix full restore ceph backups","message":"rbd: fix full restore ceph backups\n\nCurrently ceph backups only can be restored to rbd volumes.\nThe aim of this patch is to be able to fully restore ceph backups\nto any backend.\n\nThis happens because the current code is generating the wrong base\nimage when performing a full restore. If we are not using a RBD\ndestination volume we need to specify the backup in the same way as\nif a source snapshot is provided. In this way _get_backup_base_name\nreturns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db73d7f55b09146c48fde85767121f6bd3c14c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db73d7f55b09146c48fde85767121f6bd3c14c9b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"db542ce11ffcd86f6928cdca1d19259639f6aba3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-01-10 20:23:10.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/11"}}},"commit":{"parents":[{"commit":"3bebea1157dac1dfc79f5d9f2b9423c13f0845a6","subject":"Merge \"fix \u0027huawei dorado v6\u0027 create new volume from snapshot error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bebea1157dac1dfc79f5d9f2b9423c13f0845a6"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-01-10 20:23:10.000000000","tz":0},"subject":"rbd: fix full restore ceph backups","message":"rbd: fix full restore ceph backups\n\nCurrently ceph backups only can be restored to rbd volumes.\nThe aim of this patch is to be able to fully restore ceph backups\nto any backend.\n\nThis happens because the current code is generating the wrong base\nimage when performing a full restore. If we are not using a RBD\ndestination volume we need to specify the backup in the same way as\nif a source snapshot is provided. In this way _get_backup_base_name\nreturns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db542ce11ffcd86f6928cdca1d19259639f6aba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db542ce11ffcd86f6928cdca1d19259639f6aba3"}]},"branch":"refs/heads/master","description":"Rebase"},"0188c9be561b29bdc4ea25b9da7041e8ce208de6":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2022-03-11 15:23:01.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/12"}}},"commit":{"parents":[{"commit":"3bebea1157dac1dfc79f5d9f2b9423c13f0845a6","subject":"Merge \"fix \u0027huawei dorado v6\u0027 create new volume from snapshot error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bebea1157dac1dfc79f5d9f2b9423c13f0845a6"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-03-11 15:23:00.000000000","tz":0},"subject":"rbd: fix full restore ceph backups","message":"rbd: fix full restore ceph backups\n\nCurrently ceph backups only can be restored to rbd volumes.\nThe aim of this patch is to be able to fully restore ceph backups\nto any backend.\n\nThis happens because the current code is generating the wrong base\nimage name when performing a full restore. If we are not using an RBD\ndestination volume we need to specify the backup in the same way as\nif a source snapshot is provided. In this way _get_backup_base_name\nreturns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0188c9be561b29bdc4ea25b9da7041e8ce208de6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0188c9be561b29bdc4ea25b9da7041e8ce208de6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"11cf6e4cd3c86ec5bea88f72f62e54593b30fd22":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-03-29 00:32:09.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/13"}}},"commit":{"parents":[{"commit":"7de009e2a751e5ac3ca2419a38100d7f26b7a615","subject":"Add multibackend-ceph-lvm with c-bak  CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7de009e2a751e5ac3ca2419a38100d7f26b7a615"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-03-29 00:32:09.000000000","tz":0},"subject":"rbd: fix full restore ceph backups","message":"rbd: fix full restore ceph backups\n\nCurrently ceph backups only can be restored to rbd volumes.\nThe aim of this patch is to be able to fully restore ceph backups\nto any backend.\n\nThis happens because the current code is generating the wrong base\nimage name when performing a full restore. If we are not using an RBD\ndestination volume we need to specify the backup in the same way as\nif a source snapshot is provided. In this way _get_backup_base_name\nreturns backup.service_metadata.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11cf6e4cd3c86ec5bea88f72f62e54593b30fd22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11cf6e4cd3c86ec5bea88f72f62e54593b30fd22"}]},"branch":"refs/heads/master","description":"Rebase"},"fb43b548ca94637eb82ab4e50ca69eb233ce136e":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2022-03-29 19:23:42.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/14"}}},"commit":{"parents":[{"commit":"7de009e2a751e5ac3ca2419a38100d7f26b7a615","subject":"Add multibackend-ceph-lvm with c-bak  CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7de009e2a751e5ac3ca2419a38100d7f26b7a615"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-03-29 19:23:42.000000000","tz":0},"subject":"rbd: fix full restore ceph backups","message":"rbd: fix full restore ceph backups\n\nCurrently ceph backups only can be restored to rbd volumes.\nThe aim of this patch is to be able to fully restore ceph backups\nto any backend.\n\nThis happens because the current code is generating the wrong base\nimage name when performing a full restore. If we are not using an RBD\ndestination volume we need to specify the backup in the same way as\nif a source snapshot is provided. In this way _get_backup_base_name\nreturns backup.service_metadata.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb43b548ca94637eb82ab4e50ca69eb233ce136e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb43b548ca94637eb82ab4e50ca69eb233ce136e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"855053b13530068fa50c3aae847ab5524a65d243":{"kind":"NO_CHANGE","_number":15,"created":"2022-03-31 13:09:14.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/15"}}},"commit":{"parents":[{"commit":"2bbe5bba9ad7b1c646bfebd255a32aea4f381b66","subject":"Add multibackend-ceph-lvm with c-bak  CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bbe5bba9ad7b1c646bfebd255a32aea4f381b66"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-03-31 13:09:14.000000000","tz":0},"subject":"rbd: fix full restore ceph backups","message":"rbd: fix full restore ceph backups\n\nCurrently ceph backups only can be restored to rbd volumes.\nThe aim of this patch is to be able to fully restore ceph backups\nto any backend.\n\nThis happens because the current code is generating the wrong base\nimage name when performing a full restore. If we are not using an RBD\ndestination volume we need to specify the backup in the same way as\nif a source snapshot is provided. In this way _get_backup_base_name\nreturns backup.service_metadata.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/855053b13530068fa50c3aae847ab5524a65d243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/855053b13530068fa50c3aae847ab5524a65d243"}]},"branch":"refs/heads/master","description":"Rebase"},"3ad2095e65d8a90d14950ca92fc14804d3123200":{"kind":"REWORK","_number":16,"created":"2022-04-22 14:41:39.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/16"}}},"commit":{"parents":[{"commit":"614ce9e603861f1aeb67da6fd4dcb678f5f649d6","subject":"Add multibackend-ceph-lvm with c-bak  CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/614ce9e603861f1aeb67da6fd4dcb678f5f649d6"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"enriquetaso","email":"lsofia.enriquez@gmail.com","date":"2022-04-22 14:41:11.000000000","tz":0},"subject":"rbd: fix full restore ceph backups","message":"rbd: fix full restore ceph backups\n\nCurrently ceph backups only can be restored to rbd volumes.\nThe aim of this patch is to be able to fully restore ceph backups\nto any backend.\n\nThis happens because the current code is generating the wrong base\nimage name when performing a full restore. If we are not using an RBD\ndestination volume we need to specify the backup in the same way as\nif a source snapshot is provided. In this way _get_backup_base_name\nreturns backup.service_metadata.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ad2095e65d8a90d14950ca92fc14804d3123200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ad2095e65d8a90d14950ca92fc14804d3123200"}]},"branch":"refs/heads/master"},"6afd38ccb6ef99fcd90d081304697a6e3397125d":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-04-25 13:22:56.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/17"}}},"commit":{"parents":[{"commit":"4fe1a798044b6a5fecef3f2b18d625d6ef51693b","subject":"Rename and update job cinder-tempest-lvm-multibackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fe1a798044b6a5fecef3f2b18d625d6ef51693b"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-04-25 13:22:56.000000000","tz":0},"subject":"rbd: fix full restore ceph backups","message":"rbd: fix full restore ceph backups\n\nCurrently ceph backups only can be restored to rbd volumes.\nThe aim of this patch is to be able to fully restore ceph backups\nto any backend.\n\nThis happens because the current code is generating the wrong base\nimage name when performing a full restore. If we are not using an RBD\ndestination volume we need to specify the backup in the same way as\nif a source snapshot is provided. In this way _get_backup_base_name\nreturns backup.service_metadata.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6afd38ccb6ef99fcd90d081304697a6e3397125d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6afd38ccb6ef99fcd90d081304697a6e3397125d"}]},"branch":"refs/heads/master","description":"Rebase"},"822d7bf608e53e0162f73866d74a10cdc5feb07d":{"kind":"REWORK","_number":18,"created":"2022-04-25 18:01:54.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/18"}}},"commit":{"parents":[{"commit":"4fe1a798044b6a5fecef3f2b18d625d6ef51693b","subject":"Rename and update job cinder-tempest-lvm-multibackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fe1a798044b6a5fecef3f2b18d625d6ef51693b"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"enriquetaso","email":"lsofia.enriquez@gmail.com","date":"2022-04-25 18:00:12.000000000","tz":0},"subject":"[RBD] Fix Restore Full Backups","message":"[RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/822d7bf608e53e0162f73866d74a10cdc5feb07d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/822d7bf608e53e0162f73866d74a10cdc5feb07d"}]},"branch":"refs/heads/master"},"33b6b171201b41f3b4874013b2d5649bcedea654":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-04-25 18:51:02.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/19"}}},"commit":{"parents":[{"commit":"8e66c40a408f8f0e138db7871891290cb683a869","subject":"Rename and update job cinder-tempest-lvm-multibackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e66c40a408f8f0e138db7871891290cb683a869"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-04-25 18:51:02.000000000","tz":0},"subject":"[RBD] Fix Restore Full Backups","message":"[RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33b6b171201b41f3b4874013b2d5649bcedea654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33b6b171201b41f3b4874013b2d5649bcedea654"}]},"branch":"refs/heads/master","description":"Rebase"},"40553f3661a2b45fcfb9a8fc9369df0eb2a3b23d":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-04-28 19:50:54.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/20"}}},"commit":{"parents":[{"commit":"a42c0750ecd0b518223b14430c5341753a062dff","subject":"Add CI Job For lvm+ceph With c-bak Enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a42c0750ecd0b518223b14430c5341753a062dff"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-04-28 19:50:54.000000000","tz":0},"subject":"[RBD] Fix Restore Full Backups","message":"[RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40553f3661a2b45fcfb9a8fc9369df0eb2a3b23d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40553f3661a2b45fcfb9a8fc9369df0eb2a3b23d"}]},"branch":"refs/heads/master","description":"Rebase"},"eb68c0fb7abe6e9c4ba63dbe98fa04e0890cc4e4":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-04-29 13:11:08.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/21"}}},"commit":{"parents":[{"commit":"25e2258f1b68f906c03510b554bbb507d68ec1ed","subject":"Add CI Job For lvm+ceph With c-bak Enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25e2258f1b68f906c03510b554bbb507d68ec1ed"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-04-29 13:11:08.000000000","tz":0},"subject":"[RBD] Fix Restore Full Backups","message":"[RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb68c0fb7abe6e9c4ba63dbe98fa04e0890cc4e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb68c0fb7abe6e9c4ba63dbe98fa04e0890cc4e4"}]},"branch":"refs/heads/master","description":"Rebase"},"f8de943a2b996ca52ba2789a18abda3dad6f9f0c":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2022-05-16 16:58:49.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/22"}}},"commit":{"parents":[{"commit":"ca12711ffea35bd4a730f5cc6fdd41f0d84ae59b","subject":"Add CI Job For lvm+ceph With c-bak Enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca12711ffea35bd4a730f5cc6fdd41f0d84ae59b"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-05-16 16:58:49.000000000","tz":0},"subject":"[RBD] Fix Restore Full Backups","message":"[RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8de943a2b996ca52ba2789a18abda3dad6f9f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8de943a2b996ca52ba2789a18abda3dad6f9f0c"}]},"branch":"refs/heads/master","description":"Rebase"},"57e814ac20f58007ce9e6e70f1e4988f8bd71e3e":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2022-05-23 17:09:37.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/23"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-05-23 17:09:37.000000000","tz":0},"subject":"[RBD] Fix Restore Full Backups","message":"[RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57e814ac20f58007ce9e6e70f1e4988f8bd71e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57e814ac20f58007ce9e6e70f1e4988f8bd71e3e"}]},"branch":"refs/heads/master","description":"Rebase"},"7e028e953c921f9d6ebda99a90efa94f71e20e9e":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2022-11-15 11:29:02.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/24"}}},"commit":{"parents":[{"commit":"dcec2f6f01ffc63dc058f641370e9f5a0bad07e4","subject":"Merge \"Add doc8 check to docs builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcec2f6f01ffc63dc058f641370e9f5a0bad07e4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-11-15 11:29:02.000000000","tz":0},"subject":"[RBD] Fix Restore Full Backups","message":"[RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e028e953c921f9d6ebda99a90efa94f71e20e9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e028e953c921f9d6ebda99a90efa94f71e20e9e"}]},"branch":"refs/heads/master","description":"Rebase"},"48b69da1ca4810f32380fc2b5be193e1df5e3e88":{"kind":"REWORK","_number":25,"created":"2022-11-22 16:00:41.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/25"}}},"commit":{"parents":[{"commit":"dcec2f6f01ffc63dc058f641370e9f5a0bad07e4","subject":"Merge \"Add doc8 check to docs builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcec2f6f01ffc63dc058f641370e9f5a0bad07e4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-11-22 16:00:33.000000000","tz":0},"subject":"WIP: [RBD] Fix Restore Full Backups","message":"WIP: [RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48b69da1ca4810f32380fc2b5be193e1df5e3e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48b69da1ca4810f32380fc2b5be193e1df5e3e88"}]},"branch":"refs/heads/master"},"d40f438420a5abf3628843339c08b1b3c00d29ea":{"kind":"REWORK","_number":26,"created":"2022-11-24 13:18:57.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/26"}}},"commit":{"parents":[{"commit":"dcec2f6f01ffc63dc058f641370e9f5a0bad07e4","subject":"Merge \"Add doc8 check to docs builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcec2f6f01ffc63dc058f641370e9f5a0bad07e4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-11-24 13:18:34.000000000","tz":0},"subject":"WIP: [RBD] Fix Restore Full Backups","message":"WIP: [RBD] Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to rbd volumes.\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nThe reason for this is that the current code generates the wrong base\nimage name when doing a full restore. If we are not using an RBD target\nvolume, we need to specify the backup in the same way as when a source\nsnapshot is provided. This way _get_backup_base_name returns the\nbackup.service_metadata.\n\nIn particular, if the backup was created from a snapshot\n_get_backup_base_name should be call only with the volume_id.\n\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d40f438420a5abf3628843339c08b1b3c00d29ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d40f438420a5abf3628843339c08b1b3c00d29ea"}]},"branch":"refs/heads/master"},"a2bb099cae58b6ff2760c47adc7e1b8cd8b1786b":{"kind":"REWORK","_number":27,"created":"2022-11-24 15:55:49.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/27"}}},"commit":{"parents":[{"commit":"dcec2f6f01ffc63dc058f641370e9f5a0bad07e4","subject":"Merge \"Add doc8 check to docs builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcec2f6f01ffc63dc058f641370e9f5a0bad07e4"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-11-24 15:53:38.000000000","tz":0},"subject":"RBD: Fix Restore Full Backups","message":"RBD: Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to RBD volumes. The reason\nfor this is that the current code generates the wrong image name in the\npool when doing a full restore.\n\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nHistorically the first full backup used to have a specific base name in\nthe pool. However, this functionality is currently broken and it is\ndifficult to keep track of the names in the pool.\n\nIn order to fix this, this work proposes to always generate the base\nimage using the volume uuid plus the backup uuid since it\u0027s the way most\nof the names are generated today.\n\nFinally, if Cinder is not using an RBD target volume, we need to specify\nthe backup. This way _get_backup_base_name returns the\nbackup.service_metadata. In particular, if the backup was created from a\nsnapshot _get_backup_base_name should return the parent full backup\nname.\n\nCloses-Bug: #1895035\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2bb099cae58b6ff2760c47adc7e1b8cd8b1786b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2bb099cae58b6ff2760c47adc7e1b8cd8b1786b"}]},"branch":"refs/heads/master"},"eb25e874de1cd9ead99601ea84f917c380bc9f0b":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2022-11-25 10:05:42.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/28"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-11-25 10:05:42.000000000","tz":0},"subject":"RBD: Fix Restore Full Backups","message":"RBD: Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to RBD volumes. The reason\nfor this is that the current code generates the wrong image name in the\npool when doing a full restore.\n\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nHistorically the first full backup used to have a specific base name in\nthe pool. However, this functionality is currently broken and it is\ndifficult to keep track of the names in the pool.\n\nIn order to fix this, this work proposes to always generate the base\nimage using the volume uuid plus the backup uuid since it\u0027s the way most\nof the names are generated today.\n\nFinally, if Cinder is not using an RBD target volume, we need to specify\nthe backup. This way _get_backup_base_name returns the\nbackup.service_metadata. In particular, if the backup was created from a\nsnapshot _get_backup_base_name should return the parent full backup\nname.\n\nCloses-Bug: #1895035\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb25e874de1cd9ead99601ea84f917c380bc9f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb25e874de1cd9ead99601ea84f917c380bc9f0b"}]},"branch":"refs/heads/master","description":"Rebase"},"9215c3fa3a00607e6c7935132545d5b5ead33d62":{"kind":"REWORK","_number":29,"created":"2022-12-01 13:11:30.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/29"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2022-12-01 13:11:30.000000000","tz":0},"subject":"RBD: Fix Restore Full Backups","message":"RBD: Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to RBD volumes. The reason\nfor this is that the current code generates the wrong image name in the\npool when doing a full restore.\n\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nHistorically the first full backup used to have a specific base name in\nthe pool. However, this functionality is currently broken and it is\ndifficult to keep track of the names in the pool.\n\nIn order to fix this, this work proposes to always generate the base\nimage using the volume uuid plus the backup uuid since it\u0027s the way most\nof the names are generated today.\n\nFinally, if Cinder is not using an RBD target volume, we need to specify\nthe backup. This way _get_backup_base_name returns the\nbackup.service_metadata. In particular, if the backup was created from a\nsnapshot _get_backup_base_name should return the parent full backup\nname.\n\nCloses-Bug: #1895035\nDepends-On: I6f5d5c6bf6c30045752064c666dd8892adbad643\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9215c3fa3a00607e6c7935132545d5b5ead33d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9215c3fa3a00607e6c7935132545d5b5ead33d62"}]},"branch":"refs/heads/master"},"cd5b41337ec6b748854b3938533d9cbc414327c0":{"kind":"NO_CODE_CHANGE","_number":30,"created":"2023-01-23 14:00:52.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/30"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-01-23 14:00:52.000000000","tz":0},"subject":"RBD: Fix Restore Full Backups","message":"RBD: Fix Restore Full Backups\n\nCurrently, Ceph backups can only be restored to RBD volumes. The reason\nfor this is that the current code generates the wrong image name in the\npool when doing a full restore.\n\nThe goal of this patch is to be able to fully restore Ceph backups\nfrom/to any backend.\n\nHistorically the first full backup used to have a specific base name in\nthe pool. However, this functionality is currently broken and it is\ndifficult to keep track of the names in the pool.\n\nIn order to fix this, this work proposes to always generate the base\nimage using the volume uuid plus the backup uuid since it\u0027s the way most\nof the names are generated today.\n\nFinally, if Cinder is not using an RBD target volume, we need to specify\nthe backup. This way _get_backup_base_name returns the\nbackup.service_metadata. In particular, if the backup was created from a\nsnapshot _get_backup_base_name should return the parent full backup\nname.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd5b41337ec6b748854b3938533d9cbc414327c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd5b41337ec6b748854b3938533d9cbc414327c0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4dd3519b8160b77b10291209275caa9e102a9db4":{"kind":"REWORK","_number":31,"created":"2023-02-20 14:55:04.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/31"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-02-20 14:54:56.000000000","tz":0},"subject":"RBD: Fix Restore Full Backups","message":"RBD: Fix Restore Full Backups\n\nThis patch aims to enable full restoration of Ceph backups to/from any\nbackend. Currently, the code is generating the wrong image name in the\npool when performing a full restore, making it difficult to keep track\nof the names in the pool.\n\nWhat is the `base name`? Is the name to identify the backup in the ceph\npool. Usually is volume-\u003coriginal-vol-uuid\u003e.backup-\u003cbackup-id\u003e.\n\nTo fix this, this patch proposes to always generate the base image using\nthe volume uuid and backup uuid.\n\nFinally, incremental backups should return the parent full backup name.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4dd3519b8160b77b10291209275caa9e102a9db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4dd3519b8160b77b10291209275caa9e102a9db4"}]},"branch":"refs/heads/master"},"80139cad0b08eeb525efac82e6f9d50d68962063":{"kind":"REWORK","_number":32,"created":"2023-02-21 12:08:16.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/32"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-02-21 12:08:09.000000000","tz":0},"subject":"WIP RBD: Fix Restore Full Backups","message":"WIP RBD: Fix Restore Full Backups\n\nThis patch aims to enable full restoration of Ceph backups to/from any\nbackend. Currently, the code is generating the wrong image name in the\npool when performing a full restore, making it difficult to keep track\nof the names in the pool.\n\nWhat is the `base name`? Is the name to identify the backup in the ceph\npool. Usually is volume-\u003coriginal-vol-uuid\u003e.backup-\u003cbackup-id\u003e.\n\nTo fix this, this patch proposes to always generate the base image using\nthe volume uuid and backup uuid.\n\nFinally, incremental backups should return the parent full backup name.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80139cad0b08eeb525efac82e6f9d50d68962063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80139cad0b08eeb525efac82e6f9d50d68962063"}]},"branch":"refs/heads/master"},"4e98f5447fb65f4de2dafbc0bad09d875acb578d":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2023-04-20 12:58:53.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/82/750782/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/750782/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/750782/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/750782/33"}}},"commit":{"parents":[{"commit":"2e8aff660b18d1e8f961d6de979d541a8d06f1c1","subject":"Merge \"Add fips check jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e8aff660b18d1e8f961d6de979d541a8d06f1c1"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-09 18:19:25.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 12:56:05.000000000","tz":0},"subject":"WIP RBD: Fix Restore Full Backups","message":"WIP RBD: Fix Restore Full Backups\n\nThis patch aims to enable full restoration of Ceph backups to/from any\nbackend. Currently, the code is generating the wrong image name in the\npool when performing a full restore, making it difficult to keep track\nof the names in the pool.\n\nWhat is the `base name`? Is the name to identify the backup in the ceph\npool. Usually is volume-\u003coriginal-vol-uuid\u003e.backup-\u003cbackup-id\u003e.\n\nTo fix this, this patch proposes to always generate the base image using\nthe volume uuid and backup uuid.\n\nFinally, incremental backups should return the parent full backup name.\n\nCloses-Bug: #1895035\nChange-Id: I06398428f82b3617a0e8b04d86cd6a839b143263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e98f5447fb65f4de2dafbc0bad09d875acb578d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e98f5447fb65f4de2dafbc0bad09d875acb578d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
