)]}'
{"id":"openstack%2Fcinder~752781","triplet_id":"openstack%2Fcinder~stable%2Fussuri~Ibc5fa0901ae9c0766f98d623357bade266598bff","project":"openstack/cinder","branch":"stable/ussuri","topic":"bug/1887885-stable/ussuri","hashtags":[],"change_id":"Ibc5fa0901ae9c0766f98d623357bade266598bff","subject":"NEC driver: fix a snapshot detach error","status":"MERGED","created":"2020-09-19 03:41:56.000000000","updated":"2020-10-08 01:05:55.000000000","submitted":"2020-10-08 00:57:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752781-1602118678930-8f1aa1ae","meta_rev_id":"1737ba392041ce24f7c650ef715040467077fddb","_number":752781,"virtual_id_number":752781,"owner":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-19 03:56:59.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-10-07 20:31:36.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-19 04:23:56.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-08 00:57:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-07 22:17:10.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-19 05:37:06.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-07 18:15:14.000000000","_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},{"value":0,"date":"2020-10-07 21:11:38.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-10-01 14:16:22.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-07 23:37:48.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-23 04:45:29.000000000","_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-03 07:21:30.000000000","_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-07 23:16:22.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-07 13:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-07 00:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-07 13:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-19 03:56:59.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-19 04:23:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-19 05:37:06.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-23 04:45:29.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-01 14:16:22.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-03 07:21:30.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 00:42:48.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-10-07 13:58:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-07 20:31:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 21:11:38.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-10-07 22:17:10.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 23:16:22.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-10-07 23:37:48.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 00:57:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42a2139c2195daf51541fc3a555491d7da7366e3","author":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"date":"2020-09-19 03:41:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"023e0722fbadd5cdd394ee2e99249670c3426e9b","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-19 03:56:59.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/7e98b66a177d3d8d945d8146cf36c81cbe58a5b0/index.html : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9a118ea14cdf0f116632095011b563f7e3daf9ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-19 04:23:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/81/752781/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a296bf5/a296bf5.html : FAILURE in 40m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6b87cf4e0d6d319c6deb488a625b032ee5691bff","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-19 04:45:52.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/752781/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bac9716a358bcc065e254aa1ee262f4728595bfc","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-19 05:37:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/81/752781/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/30b463a : FAILURE in 1h 54m 48s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/81/752781/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/62eb8dd : SUCCESS in 1h 11m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"834f7a5b347ea9c6a7cecc0fcf10534cc8994d35","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-19 05:44:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/752781/1/check/tempest-dsvm-full-bionic/dea754d/ : SUCCESS in 1h 38m 05s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/752781/1/check/devstack-plugin-ceph-dsvm-bionic/eb7ed48/ : SUCCESS in 1h 45m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5bf3379621ff76a342a0ca1a75ae5b3edb13b14e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 05:59:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8be886a1fec64805aab129d79566568e : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6afc8dde969e4922876ec08ffa58d9df : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/390f7d4930f44bbdbe8d1f9e9fead23d : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57bdba7b2bb242558af7aaffe7db6fdc : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159f0eeef43b408987edc3967fcc15d2 : SUCCESS in 20m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/48ff129ef8d64bc8b60cc3b8dfa06c67 : SUCCESS in 1h 13m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/526299d3c3c44edcb395a46b429f93c5 : SUCCESS in 1h 16m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3678666a0ca74dd184393d2326b8d3e6 : SUCCESS in 6m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6874bce2b9e14b889df730659a3cb92c : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2b06c17fa36424da4c9d3b41925148c : SUCCESS in 5m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a22847f95d624f779f6fcf2b299f12a0 : SUCCESS in 5m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/29a5f9bc8c874e0dac7a35a938dbbd31 : SUCCESS in 37m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b947a5b7d6b841088c989bc131238a64 : SUCCESS in 4m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/75b77ae847534c43b48d0db4e37820ea : SUCCESS in 59m 50s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/92af226f4e1f4b7791b84a9ccc292579 : SUCCESS in 1h 47m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/531e29a3c8e74058a879fdcee4537a6c : SUCCESS in 1h 24m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bfc08277209644229cf80c0c57084637 : SUCCESS in 1h 34m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e82bf5388b02445fa4346f54025f583c : SUCCESS in 1h 15m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/488b5fd6e6c44c9c8f9f5f97cbf56fa8 : SUCCESS in 2h 10m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/614dd607875342079be1e5f50be691c7 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5f40bc48ac8101e93c0d6d0bda5bbad783610d88","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-09-19 06:08:29.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/81/752781/1/check/synology-dsm-cinder-iscsi/a814efa : FAILURE in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"65710a67104dfb8c770edec921d66b11592d2d92","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-19 06:54:12.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/752781/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"db3a6343f13d5c5022b6661fcbf9cab2c6ad2334","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-19 08:20:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/752781/1/20200919145030/NEC-FC : SUCCESS in 2h 26m 56s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/752781/1/20200919145030/NEC-ISCSI : SUCCESS in 2h 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f37e68fb36b36e555281f73d8c0814d8e87cd8a6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-19 09:06:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/81/752781/1/upstream-solidfire-check/cinder-solidfire-iSCSI/28857af/ : FAILURE in 53m 30s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"eb2254201d45670ff9a12fe94cd3c9a9f60819dc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-19 09:11:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-iSCSI/1ac302d/ : SUCCESS in 36m 03s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-iSCSI-IPv6/ffea168/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-NFS/20b57d9/ : SUCCESS in 36m 18s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-NFS-IPv6/316b152/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-FCP/bc38719/ : FAILURE in 3h 10m 23s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"a7eca48809d88b1426ee234a04bd3347a9f0e226","author":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"date":"2020-09-23 00:04:04.000000000","message":"Patch Set 1:\n\nrun-NEC Cinder CI","accounts_in_message":[],"_revision_number":1},{"id":"556b62bc78a52d47a7e0053a2785201d2c8eaedb","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-09-23 04:45:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/752781/1/20200923094320/NEC-FC : SUCCESS in 3h 57m 57s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/752781/1/20200923094320/NEC-ISCSI : SUCCESS in 4h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"836d346e20d0e172786bf84296d83d5f438971bb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-01 14:16:22.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-752781-1/ns5_nfs : SUCCESS in 2h 53m 13s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-81-752781-1/ns5_iscsi : SUCCESS in 2h 50m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d8235b6137e19cf0de42c8eb59d672ceead4ad84","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-10-03 07:21:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/81/752781/1/check/dsvm-tempest-zadara-driver/17da296 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6d16add2aaf687c42e724f6d053ce5d459d3870","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-10-07 00:42:48.000000000","message":"Patch Set 1: Code-Review+2\n\nBackport looks good.  Just want to note that although this modifies the support matrix, this patch is fixing a bug in an already supported capability that just happened not to be marked correctly in the documentation.","accounts_in_message":[],"_revision_number":1},{"id":"caf0316d6602a292e8ea2ca3eb960ff4eef590ef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 13:58:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged to master (victoria), good to backport.","accounts_in_message":[],"_revision_number":1},{"id":"a7ade34343d2f0cee02ecf9474a592bc60cfdcb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 13:58:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46e52393141d77ce16873680984426aab7eed147","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 17:15:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51801487f4e04912903935ed10603eb0 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c76227dadfa45a5ab194368b4432fb9 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/622e1cd88bf34c7697bf2ea001bd4adb : SUCCESS in 9m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92308d414c3642fb85897e19c7ea63fb : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d4aad54669476bb2cc1a90e947c4af : SUCCESS in 16m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/47eeff2bd515477786f4cbc47cfc6f07 : SUCCESS in 1h 16m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/310c40f3f3b849d4acd0960cc6ecca04 : FAILURE in 1h 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7133df55043345b89da1265608caabf9 : SUCCESS in 6m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fd3a12e08f8d4beb9e6cfe5e31679855 : SUCCESS in 1h 32m 00s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f3fc64dc76c4b3aac7b8a8b8294ec3f : SUCCESS in 1h 07m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bb7ca22f8aa4cc2b68ecf72f975acfe : FAILURE in 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e5a19c03a046a5bbf87c98df0cea2af7a4141d60","author":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"date":"2020-10-07 18:15:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"634b81ddcc194684428020c264060718c3b97896","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-07 20:31:36.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/752781/1/check/tempest-dsvm-full-bionic/623ad18/ : FAILURE in 1h 28m 40s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/752781/1/check/devstack-plugin-ceph-dsvm-bionic/12491f2/ : FAILURE in 36m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5853ac1874525169fcf5381d7c854b745db8e1b3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-07 21:11:38.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/81/752781/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f1603bfed930b307b52808cb8e905bebda7106ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 22:08:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcba1a5f3c4246ea8809997be60d92a1 : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e7eb8b7cf0e4945bbb742a4b076d11c : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ccf4d7534924e82ad123f4115fa553b : SUCCESS in 10m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d555aa7b55104346b180c4afa7247cae : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ebbd27190b4e53b5b17d6da757a15c : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/8d4e36864ba346189104373a5ee6dd03 : SUCCESS in 1h 15m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/053c7301817246a4bf7a334deeeb5de9 : SUCCESS in 1h 06m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd6a81261e9c420ebd202a4590745bed : SUCCESS in 6m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cedc966264a54ebf9fd6977badddb430 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d10cc00a36a4553ab19d4cd85d26c89 : SUCCESS in 7m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c5f59f12b618498fa5194e89e17a0b7c : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/745409e57b024c308dceb4680a2e18a7 : SUCCESS in 34m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e257bdf3a0b45758a4628f8d2740e0a : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b5b53ec363f49579f934ab467fdd44c : SUCCESS in 59m 33s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/141b48a3075242caa9b0a121d9e12fcf : SUCCESS in 1h 38m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c9f9c29f00e24e8e985a4f918c746e02 : SUCCESS in 1h 22m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3c00e1131b843c394ea57d55741397f : SUCCESS in 1h 21m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/20c7f45889524fa6a070b06bb1631715 : FAILURE in 1h 04m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/344031c15f6f4b06abf60f3d2b6b4931 : SUCCESS in 2h 11m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0470a85dd4a34e338f2a54051f260ae3 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f2976de07a4712a80d7828ca9ffb10e60a62a2ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 22:08:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97b50a034d03580c89eb5f9d1157bba011320571","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-10-07 22:17:10.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/81/752781/1/check/synology-dsm-cinder-iscsi/a01dbc0 : FAILURE in 8m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b2a2b888fa3c702daac652ee1ec33beec339de48","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-10-07 23:16:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/81/752781/1/upstream-solidfire-check/cinder-solidfire-iSCSI/f321019/ : SUCCESS in 1h 07m 05s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f1891d5f5fc2ea36fc8ffd2b2e0ad8cc0c22e8d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-07 23:37:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-iSCSI/9159c50/ : SUCCESS in 38m 42s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-iSCSI-IPv6/66a502b/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-NFS/e1c6865/ : SUCCESS in 38m 01s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-NFS-IPv6/c434c9a/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/81/752781/1/upstream-check/cinder-cDOT-FCP/2ab1260/ : SUCCESS in 1h 28m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"451d9c9ef9209125bb9770a088830fecb5556893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 00:57:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f1e2c14f3ee46358ce3b05b3e86bad9 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2181d7dafd204e9dac6bf6c5260cfa6b : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be5c4915391c423fb889c962f7d9414e : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df22eb53043544cabeb9005ec0334787 : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714f9724b8654cb18794e6453b9e799c : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/a3d68bacbc4e444384f2c54bd9eb9e3f : SUCCESS in 1h 14m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a9e9e7a2582d48c2ad85d4f05b51fdb9 : SUCCESS in 1h 07m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a626dbe73f24a9fb7e0b98416d0fb79 : SUCCESS in 8m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/32e59e391c34468087f02e277e4b404b : SUCCESS in 1h 17m 37s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ec76b99c54848a4b20161aeff9de9eb : SUCCESS in 1h 04m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15502cf2d986480797074142a3ea576c : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1ebfa4768cff1cc01708e6a440f85183ecd1a09d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 00:57:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f18e1369a93c427552880524502e482134658d5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 00:59:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee5a3409a6d47bcb0eb320dc89ab51c : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58fc97369e9d4d3ebdc5ab9f274389ce : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"271962b96938295e847909f4ea885f9e5ea256e8","author":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"date":"2020-10-08 01:05:55.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 39905d5df3c350d146531e8eba41719b75ea9dff","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e98b66a177d3d8d945d8146cf36c81cbe58a5b0","revisions":{"7e98b66a177d3d8d945d8146cf36c81cbe58a5b0":{"kind":"REWORK","_number":1,"created":"2020-09-19 03:41:56.000000000","uploader":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"ref":"refs/changes/81/752781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/752781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/752781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/752781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/752781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/752781/1"}}},"commit":{"parents":[{"commit":"37a527e612f78d410dd41652a4ea18ca102e1dec","subject":"Merge \"NEC driver: fix live-migration failure with FC\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37a527e612f78d410dd41652a4ea18ca102e1dec"}]}],"author":{"name":"Naoki Saito","email":"nasaito@nec.com","date":"2020-07-17 07:56:28.000000000","tz":540},"committer":{"name":"Naoki Saito","email":"nasaito@nec.com","date":"2020-09-19 03:41:56.000000000","tz":0},"subject":"NEC driver: fix a snapshot detach error","message":"NEC driver: fix a snapshot detach error\n\nIn NEC driver, _fc_terminate_connection() refers to\nVolume.volume_attachment of an argument.\nHowever, the argument is a snapshot object when a snapshot is being\ndetached.\nThe snapshot object does not have the \"volume_attachment\" attribute.\n\nThis patch avoids referring to the \"volume_attachment\" attribute when\nthe argument is a snapshot.\n\nChange-Id: Ibc5fa0901ae9c0766f98d623357bade266598bff\nCloses-Bug: #1887885\n(cherry picked from commit 6e8dd2f0c01060dbde1a0263da053ffd1a7322f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e98b66a177d3d8d945d8146cf36c81cbe58a5b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e98b66a177d3d8d945d8146cf36c81cbe58a5b0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
