)]}'
{"id":"openstack%2Fcinder~755373","triplet_id":"openstack%2Fcinder~master~I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b","project":"openstack/cinder","branch":"master","topic":"bug/1891914","hashtags":[],"change_id":"I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b","subject":"NetApp SolidFire: Fix error on cluster workload rebalancing","status":"MERGED","created":"2020-09-30 19:07:37.000000000","updated":"2020-11-18 13:12:43.000000000","submitted":"2020-11-11 22:15:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755373-1605132958011-ddfe7261","meta_rev_id":"ab05b7582e50f421bafe339dcbc615a8d08668b5","_number":755373,"virtual_id_number":755373,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-06 08:07:09.000000000","_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 12:15:44.000000000","_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-11-11 13:31:10.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"date":"2020-11-11 20:47:21.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-11-11 21:32:40.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-06 14:28:25.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 13:51:25.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-11 22:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-06 16:08:03.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-11-11 12:36:14.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29650,"name":"Thiago Correa","email":"thiagoc@netapp.com","username":"thgcorrea"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"date":"2020-11-11 20:49:07.000000000","_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"date":"2020-11-06 11:03:17.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 05:20:19.000000000","_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 20:57:57.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 20:13:44.000000000","_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"date":"2020-11-11 21:04:17.000000000","_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},{"value":0,"date":"2020-11-06 08:51:09.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 13:30:11.000000000","_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2020-11-06 14:11:23.000000000","_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"date":"2020-11-11 20:11:55.000000000","_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 21:16:44.000000000","_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"date":"2020-11-11 12:54:07.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-12 07:00:19.000000000","post_submit":true,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 21:50:52.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-06 11:13:24.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-06 16:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-06 17:29:53.000000000","_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29650,"name":"Thiago Correa","email":"thiagoc@netapp.com","username":"thgcorrea"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-06 13:29:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-10 13:15:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29650,"name":"Thiago Correa","email":"thiagoc@netapp.com","username":"thgcorrea"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":29650,"name":"Thiago Correa","email":"thiagoc@netapp.com","username":"thgcorrea"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"value":0,"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"value":0,"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"value":0,"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":29650,"name":"Thiago Correa","email":"thiagoc@netapp.com","username":"thgcorrea"},{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 19:08:23.000000000","updated_by":{"_account_id":29650,"name":"Thiago Correa","email":"thiagoc@netapp.com","username":"thgcorrea"},"reviewer":{"_account_id":29650,"name":"Thiago Correa","email":"thiagoc@netapp.com","username":"thgcorrea"},"state":"REVIEWER"},{"updated":"2020-09-30 19:59:17.000000000","updated_by":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-03 07:21:06.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 14:57:50.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 21:56:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 09:39:33.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2020-10-13 20:33:57.000000000","updated_by":{"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},"reviewer":{"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},"state":"REVIEWER"},{"updated":"2020-10-14 18:42:29.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-15 04:44:39.000000000","updated_by":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-16 07:02:32.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-04 19:09:53.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2020-11-06 08:01:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-06 08:07:09.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-06 08:51:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-06 11:03:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-06 11:13:24.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-06 13:29:00.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-11-06 14:11:23.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"REVIEWER"},{"updated":"2020-11-06 14:28:25.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-06 16:08:03.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-11-06 17:29:53.000000000","updated_by":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"reviewer":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"state":"REVIEWER"},{"updated":"2020-11-10 13:15:15.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-11-11 12:15:44.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-11-11 12:36:14.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 12:54:07.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 13:30:11.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-11-11 13:31:10.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-11-11 13:51:25.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 20:11:55.000000000","updated_by":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 20:13:44.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"REVIEWER"},{"updated":"2020-11-11 20:47:21.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 20:49:07.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2020-11-11 20:57:57.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-11-11 21:04:17.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"REVIEWER"},{"updated":"2020-11-11 21:16:44.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"REVIEWER"},{"updated":"2020-11-11 21:32:40.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 21:50:52.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 22:15:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-12 07:00:19.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a274715c02e0f292ad3af0856347690628e054b9","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-09-30 19:07:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e84c6db18167f96e73be65e3453ab938500e902","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-09-30 19:08:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/755373/1/check/fujitsu-eternus-dx-iscsi/ef76471 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/755373/1/check/fujitsu-eternus-dx-fc/fb4daf0 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"a6c2983e65f5990abe3d124f04d54048b8d7cf7a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-30 19:09:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"fd7ead7f8c042ea0b812796e1ec79b507472f18a","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-09-30 19:24:16.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/31ac5e3de8164132d70eb713104ec9df6706ffd1/index.html : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"794d3356e1d372527fc9a97b2edcacebbc81f579","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-30 19:34:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/73/755373/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b08d07d : FAILURE in 26m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/73/755373/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/cb8c7c1 : FAILURE in 25m 56s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"d573b19f515f6da328e52c5acd6ae8ec1884c759","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-30 19:41:55.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"67ccadfcb2f4ef05af200f732f603ab818ea893d","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-30 19:44:31.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"84e3a3a020e38ad2cfc7cb0ceb2494b1a7646008","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-30 19:57:42.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/73/755373/1/check-not-vote/ext-cinder-zuul/f1218ea : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4312d6477d616edde094aab38b6cf8e97c8a7c81","author":{"_account_id":29705,"name":"Intel RSD CI","email":"rsd_ci@intel.com","username":"intel_rsd_ci","tags":["SERVICE_USER"]},"date":"2020-09-30 19:59:17.000000000","message":"Patch Set 1:\n\n* intel_rsd_ci http://52.27.155.124/cinder/rsd/755373/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"97ce776319c6f7781a08bc342f1e900d484c856f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-30 20:19:36.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"747cbb93670eff3be94092a882bf21bc99c68711","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-09-30 20:19:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/73/755373/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/3cc80fe/3cc80fe.html : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f74c5ae4f73ce48ee476a9cc535721b35f4cdc53","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-30 20:28:23.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/1/check/tempest-dsvm-full-bionic/e969fca/ : FAILURE in 1h 11m 55s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/1/check/devstack-plugin-ceph-dsvm-bionic/f75de83/ : FAILURE in 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9f08950b93e2dbed50aa1a762dd4aa10f72c307c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-09-30 21:04:28.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6a534d4782ba8ea6d7499cc323b2cb115ff1b246","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-30 22:16:45.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-1/ns5_nfs : SUCCESS in 2h 50m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-1/ns5_iscsi : SUCCESS in 3h 8m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0b9f1ab99f23a766a0757941eb21df18ac5383fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-30 22:42:44.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6e4ed1d860a5960e5b34c00c759a6974b3ed62c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 00:31:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a27b13dea934c13b7da81846cd79145 : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60eef7c577d480a859e5d5fda1eedd6 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4600864b78df4b0795a890a5e3d27122 : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1058bfc5be0a4213a980afe19a9dbf3f : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6321a904dd4aa98bac8a95e07955f5 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/f102f10970aa4b7da550faedaa8d04ed : SUCCESS in 1h 08m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f37922f7f07490b9743190ba791d60a : SUCCESS in 1h 07m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8da8ad4be70a47d9b59ca86a71024ca6 : SUCCESS in 6m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2c4fa8c7febb4e85bbb77359e59a1dd7 : SUCCESS in 11m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e1172fde6b5497bbc95f40ab5190568 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32aa129859674e379162605231ae8f37 : SUCCESS in 5m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e05c8b790aa040489ef8df7997e6595b : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e8ebee6473848e5867a4e502038df48 : SUCCESS in 1h 00m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e38c22570d2476ba01520665292d819 : FAILURE in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/405bda2008aa4630b58835f20495c73e : SUCCESS in 1h 10m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0524143444ac49e6bfe11471970b23ea : FAILURE in 1h 07m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/361d72c7c55f46c48007a36c3b8dab0f : SUCCESS in 1h 20m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c982b4eff9304a95946e1acbf24ade5d : SUCCESS in 1h 06m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cdadbf4fb4d14ca7a89130995388157d : SUCCESS in 1h 12m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2e9e217479c141438d1468b23e5dfcc1 : SUCCESS in 2h 36m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/01394ca393a14172978d388d77280208 : SUCCESS in 2h 07m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6ea0340ac0a4a7c97af2c342936d7f1 : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f46a775829430c67ff6631e88975cdd318adef35","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-10-01 00:40:27.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/73/755373/1/check/synology-dsm-cinder-iscsi/f4f203e : FAILURE in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"94ede6b5d37a64af8ed3f8cb0a7ed2db0f6cf6fc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-10-01 00:51:47.000000000","message":"Patch Set 1:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-73-755373-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"1f9a0c26a040dbcf00ae07740f98d9d110aab4db","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-01 01:15:20.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e61e10fe0c4e753580b2b727a1cd11a385e41b6e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-10-01 01:46:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/73/755373/1/upstream-solidfire-check/cinder-solidfire-iSCSI/e52ffe4/ : FAILURE in 1h 08m 52s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"ca1aa914e750a332ad240a91ff2c6635afd5befb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-01 01:51:41.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a1f2af0dc06b84ef60c8c589d5eaa31c919aa6c7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-10-01 02:07:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/755373/1/20201001084257/NEC-FC : SUCCESS in 2h 21m 47s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/755373/1/20201001084257/NEC-ISCSI : SUCCESS in 2h 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a20fac9d80d8438745ca7dea6f2bcb47cc97b310","author":{"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},"date":"2020-10-01 14:48:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a02dbef4fd5f0c6086bb5b15b9fdb2f382ac855","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-10-02 10:09:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nLook good Fernando, only a fix on release notes. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"6dd53b9f8de14e079990452c484b60d7f2285a55","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-10-02 18:28:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/73/755373/1/check/nimble-iscsi-driver-dsvm-volume/a12ca57 : SUCCESS in 1h 39m 12s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/73/755373/1/check/nimble-fc-driver-dsvm-volume/ffd96a2 : SUCCESS in 3h 00m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9d38d1a9836e1c201de797537cf1ed68879abc04","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-10-03 07:21:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/73/755373/1/check/dsvm-tempest-zadara-driver/df38699 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bc90789f45acf7ac67c3b9af26724d4c6e290fc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-10-03 17:16:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/73/755373/1/check/cinder-storpool-tempest/c404909/ : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b85c6e24bb6953759ad41298953ef1846e25bc21","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-06 11:52:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d236d537c777f74e807b317866edcbf840ac25c6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-10-06 11:52:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/755373/2/check/fujitsu-eternus-dx-iscsi/7228a84 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/755373/2/check/fujitsu-eternus-dx-fc/b9f320d : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"4362d3e05c5d539128b3e8de2fe93f82a4e5b633","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-10-06 11:53:14.000000000","message":"Patch Set 2:\n\nStarting cinder-check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6aca7b390f4ffc7df90f21f9e8475fa48d4f7d41","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-06 11:53:36.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"61b2a941e3ec9865fa66f3c9835420ab04de1ecf","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-06 12:11:00.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/fcd988eb1d370188719b66734e3221c398191a95/index.html : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f9762b739f785b353d75b8f6518113a97672a1c1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-06 12:38:12.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"879672325463ff433a5f407fd23c00267012f535","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-06 12:53:29.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3178f3a6de3c341cc2ceeab7de275c0e7bc82385","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-10-06 13:07:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/73/755373/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/38e618b/38e618b.html : SUCCESS in 1h 13m 20s","accounts_in_message":[],"_revision_number":2},{"id":"1dc35af8374a220c957c6018e2598eef9b27f5f7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-06 14:04:10.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-2/ns5_nfs : ABORTED in 2h 11m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-2/ns5_iscsi : ABORTED in 2h 10m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f6b3c62d4e618d4c28e5e9aaa5fc7828fcebbcbf","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-10-06 14:57:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/73/755373/2/ : SUCCESS in 1h 08m 05s","accounts_in_message":[],"_revision_number":2},{"id":"82f2d6d9bc6c097314f460b7dd1ac39ce761ac46","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-06 15:08:14.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/73/755373/2/check-not-vote/ext-cinder-zuul/9194652 : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":2},{"id":"175019913bad1120d16b4e715aef6afa9c51efc4","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-10-06 15:23:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)\n\nHi Fernando, thanks for working on this\n\nThe code itself LGTM. Please see some nits inline.","accounts_in_message":[],"_revision_number":2},{"id":"38e33c78a35be28fce8efa86fd01e42170b254d4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-10-06 15:47:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/73/755373/2/check/cinder-storpool-tempest/c210e59/ : SUCCESS in 1h 37m 35s","accounts_in_message":[],"_revision_number":2},{"id":"38b9a7aebed44dbd84d1d3861c6b32456c0f21bf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-06 16:12:50.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/2/check/tempest-dsvm-full-bionic/b56acce/ : FAILURE in 1h 26m 22s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/2/check/devstack-plugin-ceph-dsvm-bionic/6563ca7/ : FAILURE in 26m 54s","accounts_in_message":[],"_revision_number":2},{"id":"83e30c72c0c3cca0d34a2160a7a6648f1cec8652","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-06 17:13:55.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e664d907a4443e2bd62774dfb6990e4948d7297c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-10-06 20:13:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/755373/2/20201007025539/NEC-ISCSI : SUCCESS in 2h 09m 56s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/755373/2/20201007025539/NEC-FC : SUCCESS in 2h 17m 40s","accounts_in_message":[],"_revision_number":2},{"id":"6c395c642bc3c1475e1e2bab997bd1f08f8741c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 21:34:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9332f4d06cfd45cc94dcff764ef8b24e : SUCCESS in 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff3098f121ba488baf5a3f08c39e29a7 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e55fff1935d433fac099739d10c8bb8 : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee5fe86e2e304f21bfdb0c32fe8eb3b7 : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282ebc9a27d84604ab27aeec7a9c7ade : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/b491439f1596470095adae42e37888ae : SUCCESS in 57m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76c11b343e474c55be256eb937a48f9e : SUCCESS in 1h 16m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af694036578e4b4282519814b136a6e3 : SUCCESS in 7m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4255790356cf4061bd65c31bd407fa70 : SUCCESS in 13m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6a565f3cf5d4b31893ec765898d9c4b : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9c932bd3f0e4cdfbc8c8248ec417c32 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e5c8fc313994e6fb46ba369ce3078b9 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7cc681990144a9da5025b5b535e2b8b : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63e215f1d4624e07b21032dec1cfe2ea : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9374c4b050c14a9cb2e83076282e56ce : SUCCESS in 1h 16m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/671a0dea4ebe4901833fe1c118af6035 : FAILURE in 1h 14m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8318847170454d7e8b4ed138b3670e84 : SUCCESS in 1h 28m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ea98dd65701d4722adfd1f279f6be459 : SUCCESS in 1h 04m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/71055d2cf29f44a5a7bd4a4ba394a470 : FAILURE in 1h 45m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e49729b039964db9b765f619f0bfabc5 : FAILURE in 1h 48m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38bbf114411b448fb88a8ff22bceee04 : SUCCESS in 2h 09m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60e254189fbb4f1aa05b389ca79339bd : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8de26f2f64ddd852b819a44e1695c97c98b5deb5","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-10-06 21:47:54.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com:8832/73/755373/2/check/synology-dsm-cinder-iscsi/812d482 : FAILURE in 8m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7110f8adcce3ea4df2d435d214843da477c4e2f2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-10-06 21:56:00.000000000","message":"Patch Set 2:\n\n* cinder-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-73-755373-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"6d0e65cab270d394bbf3c1711940a6e5db878bc0","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-06 22:59:50.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"343aa362c3a2102de20f30b398f800651a7d449d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-10-06 23:40:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/73/755373/2/upstream-solidfire-check/cinder-solidfire-iSCSI/e70ae93/ : SUCCESS in 2h 04m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"4e8a23fc38a53619946736af93f6f86750d0be3b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-07 00:55:47.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-2/ns5_nfs : SUCCESS in 4h 21m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-2/ns5_iscsi : SUCCESS in 4h 8m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4d16076e9affa5ebb0c05c275769285a85aad793","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-10-07 01:31:37.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/755373-2/Seagate-iSCSI : SUCCESS in 45:23.08\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/755373-2/Seagate-FC : SUCCESS in 45:41.84","accounts_in_message":[],"_revision_number":2},{"id":"70174ca7267a9dd0d1a4c36a2167fa1d45b19cec","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-10-07 01:33:19.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/755373-2/Lenovo-iSCSI : SUCCESS in 46:11.12\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/755373-2/Lenovo-FC : SUCCESS in 41:03.05","accounts_in_message":[],"_revision_number":2},{"id":"dbfa5d79266a5c91ad5f10ebc5746044ab239a74","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-10-07 04:09:10.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpemsa-ci/755373-2/HPMSA-iSCSI : FAILURE in 1:20:54\n* HPMSA-FC http://os-logs.tristero.net/hpemsa-ci/755373-2/HPMSA-FC : FAILURE in 1:20:52","accounts_in_message":[],"_revision_number":2},{"id":"6ab71d34e31528f902f3fe4470967c516d7f756a","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-10-07 17:37:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/73/755373/2/check/nimble-iscsi-driver-dsvm-volume/b740542 : FAILURE in 2h 00m 42s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/73/755373/2/check/nimble-fc-driver-dsvm-volume/172c16b : SUCCESS in 3h 11m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0233ceb9bce0182aa870b45e921b9eaaf6456726","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2020-10-08 09:39:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ce1a8274dc68f889a7606736fbf43d3457f87de7","author":{"_account_id":28710,"name":"Jay Rubenstein","email":"jay.rubenstein@netapp.com","username":"jarbassaidai"},"date":"2020-10-13 20:33:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d7782b5cf466d61ead11ceabef2cbc723315d220","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-14 18:31:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c12941014830f0d745315a86b06628d359af2979","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-10-14 18:32:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/755373/3/check/fujitsu-eternus-dx-iscsi/17f57f3 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/755373/3/check/fujitsu-eternus-dx-fc/81c0ba6 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"9695e61892db057d28545c67a74f37b39863842f","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-14 18:32:32.000000000","message":"Patch Set 3:\n\n\u003e (5 comments)\n \u003e \n \u003e Hi Fernando, thanks for working on this\n \u003e \n \u003e The code itself LGTM. Please see some nits inline.\n\nHi Lucio, thank you for you comments.\nPlease, check if commit message and release notes is better now.","accounts_in_message":[],"_revision_number":3},{"id":"7c02c991cae548a5ac9503d93fd4bfe3198ea744","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-14 18:33:12.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ffae732fc9407ff8004a75136c1db4d0a1075398","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-10-14 18:33:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/755373/4/check/fujitsu-eternus-dx-iscsi/c6e4075 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/755373/4/check/fujitsu-eternus-dx-fc/c892cb9 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"2bd85a65ea336a17268af6b299b6755c43f9ce06","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-14 18:34:10.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9d586dff0ac9bf67c44b843a4298a5445a1cf989","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-10-14 18:34:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/755373/5/check/fujitsu-eternus-dx-iscsi/e5498de : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/755373/5/check/fujitsu-eternus-dx-fc/c78879c : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"1e4b3d768bace21a23fe05d1dfa77f6d6536b402","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-14 18:42:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c0bdc0a0cb9a43820e17b3334e8432f724d8aab4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2020-10-14 18:42:29.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/755373/6/check/fujitsu-eternus-dx-iscsi/8fc93c1 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/755373/6/check/fujitsu-eternus-dx-fc/109b609 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"47e8614dbac2c10b95ff7e819006c2dd994827c4","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-10-14 18:44:19.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM, thank you Fernando!","accounts_in_message":[],"_revision_number":6},{"id":"16edc1d968d0db13ea36a65c6960b983cd64cd17","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-14 18:59:52.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/7068b833f89e32258bc43dc42b0f7bce7155931d/index.html : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"aa254ca95e9ac821f41515802bad9782beef1685","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-14 19:29:17.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/8355df7c4afdfb2acc1f0353067c5dfe36c07ba8/index.html : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e6b5bba748a3769f794713e2c6df585c55f27fca","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-10-14 19:30:14.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/6/iscsi/ : FAILURE in 46m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/6/smb/ : FAILURE in 47m 39s","accounts_in_message":[],"_revision_number":6},{"id":"6c0a174c5b265f710b21645f3c312597670ab2ac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-10-14 19:55:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/73/755373/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cf4998c/cf4998c.html : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":6},{"id":"2e151fc1054e08d61093f605332cf8885e1b6fe4","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-14 19:57:40.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/208da36a1defc2bcbe83702c987ca6685b5510e5/index.html : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"df0b963675d5edc6197aa8fa0d93c4bbf6a514e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-14 20:15:07.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"cc6c60db932b13ee309ecffa1a0288efb48eedfa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-14 20:18:19.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/6/check/tempest-dsvm-full-bionic/bfb1b75/ : SUCCESS in 1h 27m 22s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/6/check/devstack-plugin-ceph-dsvm-bionic/911672d/ : FAILURE in 20m 45s","accounts_in_message":[],"_revision_number":6},{"id":"c779e08c7776bb37024b6ac7a6135fbcae8ea05d","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-14 20:27:05.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/f359fbf5a8d5d801bad2a752448a2e42f62a2751/index.html : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"26ffaa4fd0e0ea3638b098631156f0b22ecced1e","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-14 21:00:19.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"1de1ffc5ab685ddd3c4fb9f86438de879cba685d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-14 21:35:42.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"866e261eff1fb58a259daabf8f0b67171047d1fb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-14 21:38:00.000000000","message":"Patch Set 3:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b0c97e0ef21f2002a90cdfe649267be5e401f994","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-14 21:53:57.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"bc1fd1c11b845c8ea89acd83901ef94c26c35332","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-14 22:12:12.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"0bd456c5441a9c48496b9d497a5d3e802f258939","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-14 22:16:09.000000000","message":"Patch Set 4:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"9cd2419e7761a7a9d47df6d6942726907a5c1067","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-14 22:20:46.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"fae9d1e379a8de61b8c58c6f40f771a49fa12bb2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-14 22:30:41.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-3/ns5_nfs : FAILURE in 1h 56m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-3/ns5_iscsi : FAILURE in 1h 38m 13s","accounts_in_message":[],"_revision_number":3},{"id":"1ac65ba2d6a3714e108849d2e1526cef9befe9fb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-14 22:49:40.000000000","message":"Patch Set 5:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"1937ececff4180cd519ab03e26cd8be862f5c98b","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-14 22:57:27.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"6e3d96fe502a524e4feeb26d320f9a390e89cb6f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-15 01:12:37.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-4/ns5_nfs : SUCCESS in 4h 38m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-4/ns5_iscsi : FAILURE in 1h 56m 26s","accounts_in_message":[],"_revision_number":4},{"id":"08f67c2d60f0aef53e4a7475422df65f7c80fe50","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-15 01:17:03.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-5/ns5_nfs : SUCCESS in 4h 36m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-5/ns5_iscsi : SUCCESS in 4h 10m 42s","accounts_in_message":[],"_revision_number":5},{"id":"781ff596fe48284d488b12a302b71fa660a592ce","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-15 01:41:16.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-6/ns5_nfs : SUCCESS in 4h 38m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-6/ns5_iscsi : SUCCESS in 5h 1m 8s","accounts_in_message":[],"_revision_number":6},{"id":"5c576c4ed73702f02c81b349648f364d2a84b49d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 02:07:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b29679e8c4af4cce92d96ea382fb58ef : SUCCESS in 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d09655cb8e49409a5b8ddb15711e05 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e53542989e12429e989e61eb1ab6fa6c : SUCCESS in 10m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9249ab03b82457b9aea4b41f9659644 : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a0778d62ba4362b2fc46a53a2661d3 : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/0f4329e38fc54de298d20fce499e85de : SUCCESS in 1h 23m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70c5552293e5422f90869f37f5ee3640 : TIMED_OUT in 2h 05m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40348a6a25824e9b8641ad204d6c0837 : SUCCESS in 12m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/628a74440706419e9fad7fcb845f050f : SUCCESS in 10m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dc2ae25b9494dc59f05051ef365d1c4 : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c9ac4bddde041bebe4d67952f2d77f5 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2e7a8fb95574fe5bf9cb79ce2e283cb : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3fc982cd17044f91878f9e0bc93d3d85 : SUCCESS in 33m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d21b83dcf17846b2a49a74484c421e75 : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e60d9a02e80345b4954bdb0c0ccee2a1 : SUCCESS in 1h 05m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/24fb35ee49de415898b4fbf80a33a64b : FAILURE in 1h 27m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb27c250090049259504a28fd26d1e86 : FAILURE in 1h 28m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/83efabbadb90486cb66830949e307217 : SUCCESS in 1h 15m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8906f770c7664169b71ac79247d9ef2b : SUCCESS in 1h 15m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/744a8e6993aa4e9e96d0a003a23456ef : SUCCESS in 1h 17m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16e3230b06a943e1a68ce5172e59d458 : SUCCESS in 2h 27m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71816505282a4640a2c1d8a771132001 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":6},{"id":"51ebce3dfc7dfa18303d4cf82fe2b74cb09a1eb7","author":{"_account_id":11611,"name":"Nimble Storage CI","email":"openstack-ci@nimblestorage.com","username":"nimbleci","tags":["SERVICE_USER"]},"date":"2020-10-15 04:44:39.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/\n\n- nimble-iscsi-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/73/755373/6/check/nimble-iscsi-driver-dsvm-volume/f69b7be : ABORTED in 2s\n- nimble-fc-driver-dsvm-volume https://cinderlogs.nimblestorage.com/openstack_ci/73/755373/6/check/nimble-fc-driver-dsvm-volume/fa787ee : ABORTED in 55s","accounts_in_message":[],"_revision_number":6},{"id":"55e0d297c3cf1a3f8c91bab9e07f52fa1f00d33c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-10-15 06:00:03.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/73/755373/6/check/DellEMC_Unity_FC/20d8a25/DellEMC_Unity_FC/789 : FAILURE in 1h 09m 33s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/73/755373/6/check/DellEMC_Unity_iSCSI/cffeae3/DellEMC_Unity_iSCSI/754 : SUCCESS in 1h 36m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"c8da527580170cd0896ac7f43b12d0b2534edfeb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-15 07:53:00.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"f85cd6cc2d5f76f38035fe64e44be0362a6935cf","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-15 08:36:31.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"9774eda156e3c08008ceea9aa4789772906333ee","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-10-15 11:20:52.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"defba4cb0723d8485200d1a2e3082161c9a3ccc5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-15 12:05:02.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"b7ae6e46c54da1816a1113af64c9fe55d444ed4a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-10-15 12:49:11.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/6/iscsi/ : FAILURE in 1h 03m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/6/smb/ : FAILURE in 53m 07s","accounts_in_message":[],"_revision_number":6},{"id":"6bd59759da90ab924eacc786e903239695c6baae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-15 13:06:31.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"36af60772ed20ccab5efcbd72ebc5ac6101367ab","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-15 13:27:24.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"dc114947fe629f1818c18ca2b0179efce2c2b172","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-15 13:34:24.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1d14094d4e1d3e6ef7ff9b3ae06206724da1e165","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-10-15 14:32:15.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/6/iscsi/ : FAILURE in 50m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/6/smb/ : FAILURE in 57m 26s","accounts_in_message":[],"_revision_number":6},{"id":"d505a476151cd3d8df58b31dc480b97d5f34723c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-15 14:37:20.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"280d91f2ea2397cba15632eb1ff675a80a1d7895","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-15 14:45:54.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/6/check/tempest-dsvm-full-focal-py3/8ae97eb/ : SUCCESS in 1h 20m 55s\n- devstack-plugin-ceph-dsvm-focal https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/6/check/devstack-plugin-ceph-dsvm-focal/2043f2e/ : FAILURE in 25m 54s","accounts_in_message":[],"_revision_number":6},{"id":"c633c71342315d4012f7ebcc51c3336f980b0eb0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-10-15 15:24:29.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/73/755373/6/check/DellEMC_XtremIO_FC/c25722c/DellEMC_XtremIO_FC/1109 : SUCCESS in 1h 44m 53s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/73/755373/6/check/DellEMC_XtremIO_iSCSI/6788249/DellEMC_XtremIO_iSCSI/1069 : SUCCESS in 1h 34m 38s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"c6aa127bd6ab919b1ae66a05878d74e7332f18df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-15 17:14:10.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"e46a30e489d2c7b4bcf08424d952f2bbb0497830","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-15 17:14:57.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"b625ea47a64705b9bb1b2660ef88c409eacaed99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 17:25:04.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32df6781f8d94e00bfdbd3cc5c4474a2 : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f670972a60aa47dc983d5bafc9d9bf7e : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f205e82614c74dc3b929546542fe9014 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c9a9ba378fd48cfb462d39692aa5bd7 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7127cbcbf53b4ea58954a48ca2a0a73c : SUCCESS in 14m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/6ed8d0a13543436b86c2e44e3d877d4a : SUCCESS in 1h 03m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a166d541e0d747989a0a70a4030635ac : SUCCESS in 1h 13m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad5255d69c494cf6ab6e4d9dbbabaa7e : SUCCESS in 6m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f53026766f514532859bbe548d34da3a : SUCCESS in 12m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/25764620d115423aba113941db47a863 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8da0974ed56a4587b7bdbda662142af3 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd04aa2250bc496483d5383c2555f846 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9dd968363c6048eb92560c725d3512f3 : SUCCESS in 46m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fc88e75ca9994dcd8d8df946a4a6f867 : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/919c774f681f46d5960e7d7c10164a8d : SUCCESS in 1h 28m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2de92e83455b40db9c4f75bcc0eecb1d : FAILURE in 1h 17m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71ce903951f44d46b697df5d5bcf9c93 : FAILURE in 1h 45m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/43cb399158f540cd917542cdb2fbe6d8 : SUCCESS in 1h 10m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3575999d97fe441ba6d75124026d4391 : SUCCESS in 1h 21m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fcfad685902040408dc16aeee6962903 : SUCCESS in 1h 19m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0203e449998e4261ac40b7be5e45b3b3 : SUCCESS in 2h 15m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34a0f754ec1142449101cce52b379dde : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":6},{"id":"727bb91a60d1a48bc481ffeaf46cdeb0c2ceb0a3","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-15 18:21:20.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"dfc3130d44286f07b2f267d7055830c72a287feb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-15 19:09:12.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/6/check/devstack-plugin-ceph-dsvm-bionic-py3/5d30088/ : SUCCESS in 1h 32m 20s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/6/check/tempest-dsvm-full-focal-py3/838bd9d/ : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":6},{"id":"87e86c762e206cf53ed46587f2cbbc67bf89774f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-10-15 19:34:13.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/6/iscsi/ : FAILURE in 46m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/6/smb/ : FAILURE in 47m 04s","accounts_in_message":[],"_revision_number":6},{"id":"57034b6f8ab8f67e37b34a7b5c9f82dce692b343","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-15 21:00:36.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"20832a7c2e3c7bca99fef3dedf41fe6a73579fcd","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-15 22:13:55.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"9773483488bc78986668f4993a513ec138d8837d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 23:04:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17aa51d5e5d8421db675989a4bdc6cbb : SUCCESS in 14m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1535deb8b4ea48e084eea7b69ecebe63 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/870bf1b3729a4c0299a2258af2a86021 : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/056e5ea28cf845cdb2df57c4f7728023 : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18333f7ba2bb403ebc874962556b9f37 : SUCCESS in 15m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/2934b3c653fe40298b42b5f0bc540752 : SUCCESS in 1h 06m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/514a73f43afd44fe9e2ed094a8821b4e : SUCCESS in 1h 09m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a40b3a9822b84ec08f113911a1d3ee58 : SUCCESS in 8m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/004e7adf10da4da18247cd1b28bf5847 : SUCCESS in 14m 05s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb8a624e2b9d449db1ef9366b68091f0 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d30ea4733cd74fd481b93f576b3f0e8b : SUCCESS in 7m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b16d6fbefe0347adad9adb010957c137 : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/93265e69573a44f38bad64f3619097c7 : SUCCESS in 47m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a8fa6e1321d44228878de970244df15 : FAILURE in 7m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d33aacc6608c43c4845f5282142aa572 : SUCCESS in 1h 05m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/39786c958b86484abdd0a16b240987bf : FAILURE in 1h 10m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1974882cb2644e539724d85ec4307234 : SUCCESS in 1h 37m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/14fb6888157548639ecd457265a7daef : SUCCESS in 1h 14m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/80a27eecd5a14594b02262dcfd492645 : SUCCESS in 1h 12m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/85d2dc4d521c4957849ddeb596d7a763 : SUCCESS in 1h 19m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3190c63f1744f35aa43a1bfce5f576d : SUCCESS in 2h 10m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88ad54b4105949498f80cf41bd8b8145 : SUCCESS in 1h 03m 57s","accounts_in_message":[],"_revision_number":6},{"id":"1dde4d1fd2cdd46b12ba3229491aa3905833e03c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-10-16 00:09:04.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/73/755373/6/check/DellEMC_PowerStore_iSCSI/0d49dcf/DellEMC_PowerStore_iSCSI/340 : SUCCESS in 1h 01m 39s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/73/755373/6/check/DellEMC_PowerStore_FC/67085b2/DellEMC_PowerStore_FC/337 : SUCCESS in 1h 01m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"fd033911ddd4cab0f81e4ff861f7390009c4f575","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-10-16 00:18:01.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/73/755373/6/upstream-solidfire-check/cinder-solidfire-iSCSI/efcb764/ : SUCCESS in 1h 06m 59s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"66f49bc7c909adb29b936c06daba6bf94120aecf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-10-16 02:07:49.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/73/755373/6/check/DellEMC_PowerFlex/893d8e8/DellEMC_PowerFlex/623 : SUCCESS in 2h 08m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"6387c4257933cb6af7e318857f129380bbfc09e1","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-16 02:43:17.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"713143a131d392420b043a7995554ccba8d767b0","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-10-16 03:06:53.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks Fernando, code looks good, CI is green too. LGTM.","accounts_in_message":[],"_revision_number":6},{"id":"5c7fd80fc428872e0f7f1a1a5e9dc39b46a46729","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-16 03:24:21.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"d7c6d38449d50de0dab06ae047c4d197a870682c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-10-16 04:20:04.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/73/755373/6/check/DellEMC_Unity_FC/01546e2/DellEMC_Unity_FC/803 : SUCCESS in 1h 12m 19s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/73/755373/6/check/DellEMC_Unity_iSCSI/13669cf/DellEMC_Unity_iSCSI/768 : FAILURE in 1h 39m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"4ad45c3c30a7c51e3ecb9372b1d7e7b56d841980","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-16 05:08:53.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"4aaad33367cb4c85a936b2a4a620b0812fe48709","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-16 05:35:32.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"60e7e5487313fa7274a5af473e180de31de0ba13","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2020-10-16 07:02:32.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/logs/73/755373/6/check/cinder-storpool-tempest/82f96a8/ : SUCCESS in 1h 32m 15s","accounts_in_message":[],"_revision_number":6},{"id":"2cc5e924351b31a13191ec3b67a8dbbf2043f6ea","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-10-16 09:44:34.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/73/755373/6/check/DellEMC_XtremIO_FC/892c0f2/DellEMC_XtremIO_FC/1120 : SUCCESS in 1h 40m 57s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/73/755373/6/check/DellEMC_XtremIO_iSCSI/3e1ec86/DellEMC_XtremIO_iSCSI/1080 : SUCCESS in 1h 38m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"5d69451fc6dadf035d42c5b174f85cfca2618360","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-16 10:20:52.000000000","message":"Patch Set 6:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"45f1710661d3009c84a30ec2b4fe9de65ea5adea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-16 10:59:48.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"5e1d28cc590a3a9ec330c4005fe362d11c0aa3ff","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-10-17 02:12:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/755373/6/20201017075809/NEC-ISCSI : SUCCESS in 3h 05m 26s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/755373/6/20201017075809/NEC-FC : SUCCESS in 3h 14m 16s","accounts_in_message":[],"_revision_number":6},{"id":"2a62e521b81b24649f1305535c5a0665166248bf","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-28 07:48:20.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/73/755373/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/6f638ab : FAILURE in 10s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/73/755373/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/de35ce3 : FAILURE in 20m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"1629254ea18c892cf3a7e55832b8c26dd3ad6009","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-10-29 21:55:14.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nNit about the release note.  Code change looks fine.","accounts_in_message":[],"_revision_number":6},{"id":"c034d403c0904c3b284a2cca623418b969b942ae","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-30 14:40:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7ee59569d01aa438cb2ad5b10b673e6d7c9ef91a","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-10-30 14:40:44.000000000","message":"Patch Set 7:\n\nBuild Unstable \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/f241072c38a2063897f085595565c40789acea6b/index.html : UNSTABLE","accounts_in_message":[],"_revision_number":7},{"id":"3c889073e247791c1458506b6efe7bbc4993ffb3","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-30 14:50:55.000000000","message":"Patch Set 6:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e Nit about the release note.  Code change looks fine.\n\nThanks Rosmaita, I\u0027ve changed the reno as recommended.","accounts_in_message":[],"_revision_number":6},{"id":"84fc6cf15022ad19702d1b6faa5fd1b743d6b970","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-10-30 14:52:22.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n \u003e \n \u003e Nit about the release note.  Code change looks fine.\n\nThanks Rosmaita! I\u0027ve fixed the reno as recommended.","accounts_in_message":[],"_revision_number":6},{"id":"6d20912450da4cfd6df61e3395bca8c24a91baa8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-10-30 14:54:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/73/755373/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d70c114/d70c114.html : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":7},{"id":"d5f067df1e2dfcda7f83d2a62916554340ebea90","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-30 15:01:25.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/73/755373/7/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/0862848 : FAILURE in 20m 40s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/73/755373/7/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/9cbe435 : FAILURE in 19m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":7},{"id":"4a8bf4f546f557cc753e1bf70d7958358432e54e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-10-30 15:31:42.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/7/iscsi/ : FAILURE in 49m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/7/smb/ : FAILURE in 50m 49s","accounts_in_message":[],"_revision_number":7},{"id":"131bdde17602100a6afca0e668009f86fdfb9297","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-30 16:14:58.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"92e85e25f64529f50cf1a5f5103409c11fa42303","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-10-30 17:20:23.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/7/check/tempest-dsvm-full-focal-py3/7dcf8e7/ : SUCCESS in 1h 22m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/7/check/devstack-plugin-ceph-dsvm-focal-py3/4fd8414/ : SUCCESS in 1h 25m 49s","accounts_in_message":[],"_revision_number":7},{"id":"619a20cfa720203f6f3c81ef7cde70067236e035","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-30 17:25:11.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"4df01b5c8f8bd2a1821b58c43401be3dfc38e7fe","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-30 17:28:24.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"1eaffff1d07040795a2f96ad217c6d135eb1c828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 00:03:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd9219a7277046d5adb330de25057632 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d232d42f6248dfab277e03746d0cd3 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94ceaa99443542278d786d25a2906a81 : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c418d3be14449569cee23fab8a69782 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d8a356cf5a4829abfb55a560dcfa99 : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/30ad594881b3448596de61a3085962be : SUCCESS in 55m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d780ace4f8b4e228f0a5174084c6d6d : SUCCESS in 1h 18m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e07357f9778e4ab7b37e4f3db2c973c8 : SUCCESS in 8m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ef706dd13a24b8cb08fff4716ac3f8e : SUCCESS in 12m 59s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c43e6ebc05b4f3c92ec12b4c2e98ab2 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80a9fd89e5cf42799baaa0b3c782b9d2 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3ab222bc43fe47849e984fd3624df38d : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1a4ab7635d0f4dc3b506bb276bd9a943 : SUCCESS in 44m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/03375362ce7546ae99c45c42ddf5bb57 : FAILURE in 5m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51385915f55347bb9c87acc1dbbfe99d : SUCCESS in 1h 05m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e7f9083a867c4afa80b98e805f6fdb03 : FAILURE in 1h 10m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d67df5095caa4ccda76e57bc1ce2215b : SUCCESS in 1h 29m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d78950ce7f2442dfbb8d6505738f7b33 : SUCCESS in 1h 32m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e2a15bc5380486e8459a46ed15df574 : SUCCESS in 1h 37m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/94167ba0b7684a02b0ffae9dd653d60f : FAILURE in 1h 12m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3eb17f07a214410a9e203397900d7d8b : SUCCESS in 2h 29m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41fc4f908e9b44848b5bbb6264f61f84 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":7},{"id":"4e1651ae763f651db83cb337d303e48408722fc3","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-10-31 00:04:09.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-Synology DSM CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/73/755373/7/check/synology-dsm-cinder-iscsi/942a643 : FAILURE in 13s","accounts_in_message":[],"_revision_number":7},{"id":"063028d5d918665e9cf770b2a1131cb674620b9b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-31 00:08:39.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-7/ns5_nfs : SUCCESS in 7h 16m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-7/ns5_iscsi : SUCCESS in 7h 5m 54s","accounts_in_message":[],"_revision_number":7},{"id":"4a443d5695ecb9a0d064cb221001d992d2cfacf9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-10-31 01:16:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/73/755373/7/upstream-solidfire-check/cinder-solidfire-iSCSI/b3b7ecb/ : SUCCESS in 1h 09m 07s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"3bde7fc95d2f4946393873d7352b7741baa453e4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-10-31 01:51:30.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/73/755373/7/check/DellEMC_PowerStore_iSCSI/383ad23/DellEMC_PowerStore_iSCSI/44 : SUCCESS in 1h 04m 22s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/73/755373/7/check/DellEMC_PowerStore_FC/818e121/DellEMC_PowerStore_FC/40 : SUCCESS in 1h 11m 22s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"0afb0a02f498a573af84340a9eef580dcfdb0fd2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-31 02:54:21.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/73/755373/7/check-not-vote/ext-cinder-zuul/4cd08d1 : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":7},{"id":"fd30a552c7d8f064ada1f8c1f7611dbe1a37fca6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-10-31 03:40:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/73/755373/7/check/DellEMC_SC_FC/da62473/DellEMC_SC_FC/26 : FAILURE in 2h 00m 25s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/73/755373/7/check/DellEMC_SC_iSCSI/38a96b0/DellEMC_SC_iSCSI/24 : SUCCESS in 1h 23m 55s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"9c620890db1df7b5a4502fbb8861f9a5137a9050","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-10-31 03:47:37.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/755373-7/Lenovo-iSCSI : SUCCESS in 45:47.60\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/755373-7/Lenovo-FC : SUCCESS in 42:29.54","accounts_in_message":[],"_revision_number":7},{"id":"19e9fefcddefa0ce9077a6f0dd92ce2f75655514","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-10-31 03:48:55.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/755373-7/Seagate-iSCSI : SUCCESS in 47:53.21\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/755373-7/Seagate-FC : SUCCESS in 47:15.12","accounts_in_message":[],"_revision_number":7},{"id":"39a7c28253d40937aebcfa4bb03c3e73f95c7f6a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-10-31 03:52:11.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/73/755373/7/check/DellEMC_PowerFlex/82199e1/DellEMC_PowerFlex/38 : SUCCESS in 2h 08m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"ef86bf8069a768dc78823e5354416a2b0f416010","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-31 04:45:38.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"a44052b4688af0d09d6aab5e03a3e85a066dea9c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-31 04:50:51.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"66cf4ce489013ed2fdebd1b83f8ff8343ec22dbd","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-10-31 05:04:16.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpemsa-ci/755373-7/HPMSA-iSCSI : FAILURE in 1:22:35\n* HPMSA-FC http://os-logs.tristero.net/hpemsa-ci/755373-7/HPMSA-FC : SUCCESS in 36:49.95","accounts_in_message":[],"_revision_number":7},{"id":"fbfdef5110b9d0b20579d30e76f326e73f9b206f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-10-31 05:16:55.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/73/755373/7/check/DellEMC_VNX_iSCSI/cff4f8f/DellEMC_VNX_iSCSI/38 : SUCCESS in 2h 49m 23s\n- DellEMC_VNX_FC http://publiclogs.emc.com/73/755373/7/check/DellEMC_VNX_FC/89c5a21/DellEMC_VNX_FC/38 : SUCCESS in 1h 05m 53s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"40c39d310f2cc05bb4bde2568d4626ace0b944d7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-10-31 07:14:54.000000000","message":"Patch Set 7:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"83af8fa0ed95433f5a3a9f3478ce62761d757a01","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-31 07:52:35.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"5d4d68ab7a5262f27dafde38d95dd1b67b3d081f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-10-31 09:21:01.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/73/755373/7/check/dsvm-tempest-cisco-zonemanager-job/676f6cf/ : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":7},{"id":"63cc217514ed60a92c1f7918eebc07c275bc9968","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-11-03 11:58:48.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"653c086adbdb174c577d1f160f6b6afae50b9504","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2020-11-04 19:09:53.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"de4590c440251019637e87c8c33b3c5256234ea3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-11-06 08:01:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/755373/7/20201106130142/NEC-FC : SUCCESS in 2h 05m 44s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/755373/7/20201106130142/NEC-ISCSI : SUCCESS in 4h 00m 00s","accounts_in_message":[],"_revision_number":7},{"id":"2dfa6df2249501151c171ed10b312d2f92508173","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-11-06 08:06:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5595fb1e16773d8c19c1f4e32ebda60d4dcaa6c3","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-11-06 08:07:09.000000000","message":"Patch Set 8:\n\nBuild Unstable \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/8d0c08b4bb9d32f2c02dcc60f1f482331197e546/index.html : UNSTABLE","accounts_in_message":[],"_revision_number":8},{"id":"b10e777550f1de04042922c3059a594c7b499388","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-06 08:09:36.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"3fbad4f542c8dfc079797b4b6923f8fca9914add","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-11-06 08:09:47.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nThanks Walter for your review, I\u0027ve added the change to the version history and bumped version as suggested.","accounts_in_message":[],"_revision_number":8},{"id":"bfef94817f00ae9fc173a27582659766a70c30e6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-06 08:27:41.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/73/755373/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/462d00d : FAILURE in 20m 31s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/73/755373/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/3bff61b : FAILURE in 19m 56s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"b6bb4a6d49bad0c7eb291040ba85910d6dca3a00","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-06 08:31:43.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/73/755373/8/check/dsvm-tempest-cisco-zonemanager-job/699fe77/ : FAILURE in 18m 16s","accounts_in_message":[],"_revision_number":8},{"id":"a74617a64d262e7001f580fe52cf6c5a1c720abc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-06 08:34:53.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"715cc191eccd34382587b070d0e281f42da3f3aa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-06 08:50:21.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/73/755373/8/check-not-vote/ext-cinder-zuul/6e53ce8 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":8},{"id":"8d03599731f3ec0dea0f31e34d386959cd178352","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-11-06 08:51:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/73/755373/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/df1010d/df1010d.html : FAILURE in 42m 17s","accounts_in_message":[],"_revision_number":8},{"id":"fb4b0b71519d3ae4ff6c559966044f2910889d73","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-06 09:19:33.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/8/iscsi/ : SUCCESS in 1h 01m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/8/smb/ : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":8},{"id":"ac93e1c56c08bcbe51e8d57882df5a5acea76e61","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-06 09:50:30.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/8/check/tempest-dsvm-full-focal-py3/c1e8cfa/ : SUCCESS in 1h 16m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/8/check/devstack-plugin-ceph-dsvm-focal-py3/3a5b335/ : SUCCESS in 1h 20m 01s","accounts_in_message":[],"_revision_number":8},{"id":"45cfe54480c7f48f8e194d83f241097731a55fff","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-06 09:55:18.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"e8b90aad13ffd43a14e9b82af355007fa692f26a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-06 10:30:40.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"f5370279867971bb20e81d587e934f80358ae6f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:47:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f45bb32526114e50b62119c3ba5087c1 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d44a4beda0ae44eb81022b95aa3222db : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e985703f734039a46f04176508148e : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0175b6f9f75346868acedf31a743b971 : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88eb326a262e4c9c8e92c5869c6a0084 : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/7e40008bb12d466bb070f5f7958b3621 : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89e6070517654eb58a21d9b38e3568e6 : SUCCESS in 1h 04m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/458c4158657d4f05b96a352f149b7bf3 : SUCCESS in 6m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/092b99034da245f0ae0c1344cabe0829 : SUCCESS in 11m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69cd59edd6354ce391ea09cb5c241aa3 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e26561c541344a38853b454ea6d2d2de : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/29b0dc510e3d4ffda207e169a4eefec8 : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9141591a6df84d03ae2133339785d8e5 : SUCCESS in 42m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d4b1f8f8a9248f9bee486fb35997121 : FAILURE in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8ea214aa216c4d4eb35de0848baaeda7 : SUCCESS in 1h 08m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ae1ad7b446824faaa5b7ef825497f4a1 : FAILURE in 1h 16m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/413758fe22034f0da32c102cd9753a07 : SUCCESS in 1h 27m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7bbadd7acf7940efb3e94ce12d59a115 : SUCCESS in 1h 12m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/06f00154a71645c5ab0ac9d34ff12a95 : SUCCESS in 1h 14m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7f926bdb3df1434eb4e76b91a546cfcc : SUCCESS in 1h 32m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eab4c8c11f5948848d0df6150d980f21 : SUCCESS in 2h 08m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1263a1deab0418c85689192644ab53d : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":8},{"id":"d747fee014d00624e894a38f93b716029fa4ff2c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-11-06 10:53:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/73/755373/8/upstream-solidfire-check/cinder-solidfire-iSCSI/bcc62b7/ : FAILURE in 4m 41s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"4b433a23fb67cc68a63aab7e37c4c25516f3c1c8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-06 11:03:17.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-8/ns5_nfs : SUCCESS in 2h 55m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-73-755373-8/ns5_iscsi : SUCCESS in 2h 54m 51s","accounts_in_message":[],"_revision_number":8},{"id":"3fc3bfa3e0e62c2f7dcca7d3eba63d7b0d9097ae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-06 11:13:24.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/73/755373/8/check/DellEMC_Unity_FC/251f570/DellEMC_Unity_FC/99 : SUCCESS in 1h 37m 14s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/73/755373/8/check/DellEMC_Unity_iSCSI/959ec68/DellEMC_Unity_iSCSI/84 : FAILURE in 1h 41m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"3d6fcf7adb8f2aefedb5ad9ba2cd3f05c35bdd0c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-06 11:59:55.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"3a8a8f2ff5ad5db42c5b5368a4d26273e6ad7fb3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-06 12:10:37.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"40ebe46c7fc8deb5c44def976e8b3b47e832d76b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-11-06 12:15:53.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/73/755373/8/check/DellEMC_PowerStore_iSCSI/918f902/DellEMC_PowerStore_iSCSI/70 : SUCCESS in 1h 02m 30s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/73/755373/8/check/DellEMC_PowerStore_FC/becfb1f/DellEMC_PowerStore_FC/66 : SUCCESS in 1h 03m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"67908848f1f334a04280b3ad866abaac894f1b23","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-11-06 12:19:27.000000000","message":"Patch Set 8:\n\nrun-NetApp Solidfire CI","accounts_in_message":[],"_revision_number":8},{"id":"5e1086c336c7a38e124b876c1daf82a910d18960","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-06 12:47:40.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"26e76c169dd42e54750e8e5f6aa8761e99119ec5","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-06 12:52:05.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"c92caa98653523858270261fc93b468d48945ef3","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-11-06 13:23:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/73/755373/8/check/synology-dsm-cinder-iscsi/066b972 : SUCCESS in 1h 07m 19s","accounts_in_message":[],"_revision_number":8},{"id":"47689d8ff8e95549751a8acc3f3682d8a869661b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-11-06 13:26:36.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/73/755373/8/upstream-solidfire-check/cinder-solidfire-iSCSI/acd4490/ : SUCCESS in 1h 06m 37s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"c8ccbdfef25c071e8f5112c7b82451e9ac790238","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-11-06 13:29:00.000000000","message":"Patch Set 8: Code-Review+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":8},{"id":"8a6985a53ce04236a1307c118f0c78ffadd65c23","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-11-06 13:53:21.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/755373-8/Seagate-iSCSI : SUCCESS in 45:01.52\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/755373-8/Seagate-FC : SUCCESS in 37:40.75","accounts_in_message":[],"_revision_number":8},{"id":"06f90ea7d968fb6f7edd9965590b274c8e4303ae","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-11-06 14:11:23.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpemsa-ci/755373-8/HPMSA-iSCSI : SUCCESS in 53:57.26\n* HPMSA-FC http://os-logs.tristero.net/hpemsa-ci/755373-8/HPMSA-FC : SUCCESS in 37:37.61","accounts_in_message":[],"_revision_number":8},{"id":"093c4f68b3bcdbeac9cb63e91b5c80285bcbfc97","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-11-06 14:11:45.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/755373-8/Lenovo-iSCSI : SUCCESS in 47:51.62\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/755373-8/Lenovo-FC : SUCCESS in 44:56.88","accounts_in_message":[],"_revision_number":8},{"id":"c8ae3884f5fd96b9b8d9e85c96850cdbb8283c9b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-11-06 14:28:25.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/73/755373/8/check/DellEMC_PowerFlex/23b97c4/DellEMC_PowerFlex/64 : SUCCESS in 2h 10m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"efa222aec5ead17b69bdd4f81091f71f082cd936","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-06 14:56:27.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"bcfd465a268e60cdcb9604897fd04236d2708a79","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-06 15:05:19.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"adf58bd805b7c5086fa1ca71d9cea2ba8c000f62","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-11-06 16:08:03.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/73/755373/8/check/DellEMC_SC_FC/bdfd74b/DellEMC_SC_FC/52 : FAILURE in 2h 02m 15s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/73/755373/8/check/DellEMC_SC_iSCSI/0b3097e/DellEMC_SC_iSCSI/50 : FAILURE in 2h 51m 07s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"6fcfb5e0a614d5fe52e1cd7ed9ef735a0c911651","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-06 16:19:39.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/73/755373/8/check/DellEMC_VNX_iSCSI/ddef8c4/DellEMC_VNX_iSCSI/64 : SUCCESS in 2h 46m 49s\n- DellEMC_VNX_FC http://publiclogs.emc.com/73/755373/8/check/DellEMC_VNX_FC/4db4696/DellEMC_VNX_FC/64 : SUCCESS in 58m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"ea605f750eceed5e01a651ba5e092bd8dd85b7aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-06 16:33:17.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"8cf56f7c12a8f40c6e146dde4b639d9da35409e8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-11-06 16:44:51.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":8},{"id":"fbe88e0b5e474a70a10f1674dd86340fe56332aa","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-06 16:50:01.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"fe4ab61294a88f70b27ea50a80016e405660b614","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-11-06 17:29:53.000000000","message":"Patch Set 8: Code-Review+2\n\nGood catch, @Walter.\n\nLGTM.","accounts_in_message":[],"_revision_number":8},{"id":"b099a8f34703f9db38c7587c674b77234b2bb95b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-11-10 13:15:15.000000000","message":"Patch Set 8: Workflow+1\n\nI don\u0027t see any reason to hold this up.","accounts_in_message":[],"_revision_number":8},{"id":"46cab0a76be1e04c2440f97a4907da0efedcc438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 13:16:12.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2ec88ad5174289e3026c4fdffb42c021ccbacc15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 18:13:29.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49ef910c0e094c48b90e2acca2653b39 : SUCCESS in 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7b46d579ad4897a93b90685c7967a7 : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9f97ff492c14ac4b8f139bf16440302 : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e9e18fce26140a0ace7766de2bf3067 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c924c6848d9f4cda853a1314e16a8b56 : TIMED_OUT in 32m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/d98623db6bf44b11b1808401df1dc410 : SUCCESS in 59m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89024852834e4b71a685ebfb6287098b : SUCCESS in 1h 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e57d8a4e24784f0fab7ae25e2dd47145 : SUCCESS in 5m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcab16d4268a498499a147a3aac4472e : SUCCESS in 1h 23m 52s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/793acf02a4ca4cf9b4ece228bbf98aab : SUCCESS in 1h 08m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f4e01f72a8a477381a73f1fa6502841 : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":8},{"id":"44f7c38b455be42d25072a04596afce213f466bf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-11-10 21:09:00.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"7b2fdbc8eb6372cb9feaa9c5294898e8daebb484","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-10 21:29:16.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/73/755373/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4a0a043 : FAILURE in 19m 59s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/73/755373/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/2842d3f : FAILURE in 19m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"c9eb327d277ba870f2cf293f08e499b8d7cad5e3","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-10 21:44:49.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"e56a459b59b7710d4e7d5ad0e645c013f0035746","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-10 21:50:27.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/73/755373/8/check-not-vote/ext-cinder-zuul/1d7f529 : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":8},{"id":"9ef943f803e0b035ef808a89986c820160831965","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-10 22:07:20.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/8/iscsi/ : SUCCESS in 56m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/8/smb/ : FAILURE in 57m 54s","accounts_in_message":[],"_revision_number":8},{"id":"aec0d7d3c58a92af14ba043ce6461ea2cb03bd91","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-10 22:53:11.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"85c40a630491f6bfdcd56468742c6742eb316e21","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-10 23:18:33.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"bef8419f166a1588d10b2d4225e612b20af2aeec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-10 23:28:53.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/8/check/tempest-dsvm-full-focal-py3/5d0504b/ : SUCCESS in 1h 21m 38s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/8/check/devstack-plugin-ceph-dsvm-focal-py3/1b64037/ : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":8},{"id":"926d5282427d418d0fe343042c2582dc9ce64bb8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-11 00:03:04.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"824a21f10cb989587fda51d5fd8c82bf7874c283","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-11 00:34:49.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/73/755373/8/check/dsvm-tempest-cisco-zonemanager-job/1b853e1/ : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":8},{"id":"fce810b96f2bdb6bf33787e0e48190a6b6bb0119","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-11-11 05:20:19.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e601ba6c2e123d7efe8a4c361f2b01da51ae7897","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-11 05:23:01.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"bb9f87d0fc95ae5a3aa35f7b858b33f4d2bc132b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-11 05:40:37.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/73/755373/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e173a10 : FAILURE in 19m 39s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/73/755373/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/446d4b4 : FAILURE in 20m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"d412cb53cdfbf9ea075ab27f4c0062e5cae26ccc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-11 06:06:06.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/73/755373/8/check-not-vote/ext-cinder-zuul/5c5dc63 : SUCCESS in 45m 15s","accounts_in_message":[],"_revision_number":8},{"id":"c00515afd2c0672f02b8386b409cb146ba2912f3","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-11 06:21:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/8/iscsi/ : SUCCESS in 1h 00m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/8/smb/ : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":8},{"id":"573ec9b2a345c4b28241830f44cab300dccea0e3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-11 07:14:25.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/8/check/tempest-dsvm-full-focal-py3/8094b88/ : SUCCESS in 1h 09m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/8/check/devstack-plugin-ceph-dsvm-focal-py3/f39867f/ : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":8},{"id":"feccabda4308cf7d369b7c2eed610ef00bc2dde8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-11 07:47:15.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"fd5ac3a5ce888a7a26188c09d42e69b6d4243fc7","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-11 07:48:54.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"791aef78dd33a440c298f67d306d1cea59756b23","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-11 07:50:44.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/73/755373/8/check/dsvm-tempest-cisco-zonemanager-job/3c89317/ : FAILURE in 1h 10m 28s","accounts_in_message":[],"_revision_number":8},{"id":"6f6fceb71d78a98492065ee75d51ef7538f10058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 11:50:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cd03da820fd4f27861d8db38019ddb6 : SUCCESS in 19m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2e4bd6113cf41b38af3223830016b54 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d278968e7c48459fbf1110025f89a888 : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/648aca04fb274d8faf2120d0c3603a9e : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10413f61d984cefbe012bda2b16805d : SUCCESS in 14m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/961383b0855c4ac6bdcc1e872ea33fb7 : SUCCESS in 1h 07m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6febcd2b3bfb4df8ae9003d43a6df190 : SUCCESS in 1h 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da5480277b6a456ab375827c7aa49d33 : SUCCESS in 7m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e97e782aef644e0aaf936ba6e0047986 : SUCCESS in 11m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/27a30a9e0f844eeca59e2665fbfefcdb : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a764493f0d944ce489fe6c6571786f2e : SUCCESS in 6m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2770861af39c4389aa2c3dbd2f25836a : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b27b85cabe1944c18ac2b59df5f481f9 : SUCCESS in 47m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7f1d872c4f0d4724b1f03e62ba6a89cc : FAILURE in 4m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e6171311aadc49b8b58b84d4eeb8e943 : SUCCESS in 1h 00m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f36e1542b6bc47f282ff84b154983e34 : FAILURE in 1h 12m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dd8fc6b63b0e45818ac79a2c524cfe2e : FAILURE in 1h 28m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/46b93fdf93964477807e90ac7a4105d2 : SUCCESS in 1h 13m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fb29947140984818bf0869d7d61b9c2a : FAILURE in 1h 17m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab039c84f4124e83ac8997320033bdf2 : SUCCESS in 1h 33m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2998739c456f417997074ff05f5fea93 : SUCCESS in 2h 07m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9755999d10844e6bb08300cbe3781b10 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":8},{"id":"7ec8874a5c20caded83a79ee0eb2b075a1b05dc1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-11-11 12:15:44.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"a628f4dd81f0497979d9884b6e97cf41fde9fbb6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-11 12:36:14.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/73/755373/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/f1ef65d : FAILURE in 20m 11s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/73/755373/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f098893 : FAILURE in 20m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"b77fcc75b234ce283b80fb1e98665fc8d1d5f5b8","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-11 12:50:06.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"021916a41393a48015e006fa85d1b4bcfc09825f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-11 12:54:07.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/73/755373/8/check-not-vote/ext-cinder-zuul/3f7897f : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":8},{"id":"5f7d583fd4d0123c96e6d30c23fd3ea5b0d59e07","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-11 13:30:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/755373/8/iscsi/ : SUCCESS in 1h 01m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/755373/8/smb/ : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":8},{"id":"35d697617877804080525f5ead14cd7322019b84","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-11 13:31:10.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/73/755373/8/check/dsvm-tempest-cisco-zonemanager-job/6dd921f/ : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":8},{"id":"92293219f58cf2a197e13b35bff5eb45efa9b7e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-11 13:42:28.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"13282a746435821cf435bd0619c39e97dc7d12c6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-11 13:51:25.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/8/check/tempest-dsvm-full-focal-py3/f9cd037/ : SUCCESS in 1h 16m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/755373/8/check/devstack-plugin-ceph-dsvm-focal-py3/868f835/ : SUCCESS in 1h 26m 06s","accounts_in_message":[],"_revision_number":8},{"id":"5effc30e5cc7d8be8039941c7a8c0c7fab85b31c","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-11 14:28:00.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"34b336a7f9b1c4708a20324be8ef6774166eb448","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-11 15:14:02.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"7445f0754fa42f6fa83c0c7f24caa60837fb9299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 19:06:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea4f83ca03f44d42888a9d0d9d627a7f : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf1d85088f144669201b4b4804c5561 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88f77ed1aec74f26ba3f0dac0ed35d4e : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2759802dbb754c4ba00c531c7f7a58e7 : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c475236fba7e40faad04017f815520ec : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/e2fe9dc7a56a4e4f8653342ad8126ad1 : SUCCESS in 1h 04m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b5e3fc31bf54166987a150e6b3df42c : SUCCESS in 1h 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a09a04e1797f4a8b8dbaf5b18094d82b : SUCCESS in 7m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/55c0aad63d4e4504b193363dc06d77c8 : SUCCESS in 20m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79ee8f048c3e4710b299b64d57d05ec9 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ea1ef3501064d6fb6433f6e7672b6b5 : SUCCESS in 5m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77037451d9b04f29ba85454706e2e773 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ec6bbd95871143af877ca605d2810539 : SUCCESS in 57m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d32a7867a56f41f8aecdc2876c482d18 : FAILURE in 4m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/77223c17b8f44a9fae3a4d67202ac8c3 : SUCCESS in 1h 10m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/18c101f9f7f74ef88c22937a9fef75f7 : FAILURE in 1h 21m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a09e89362eb4dbdbb2e83b3caa685bf : SUCCESS in 1h 23m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3d6a8769f8184791900c73be5e89c65b : SUCCESS in 1h 41m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/84a645f8bc0f4ccf8852a25e3f1c682d : SUCCESS in 1h 26m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6d4cda15009a40738db9ba62a8774a1e : SUCCESS in 1h 28m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4ea6853338e945a3a236ff3d71433fbe : SUCCESS in 2h 01m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/646677e0a3084c81ac620f34f7914d53 : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":8},{"id":"1152c7015045bed3a95b8499caa637172537a039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 19:07:48.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0bf111bbcdfa4b0485a2eb969f698d7b37e6ccaa","author":{"_account_id":22510,"name":"Synology DSM CI","email":"cinder@synology.com","username":"synocinder","tags":["SERVICE_USER"]},"date":"2020-11-11 20:11:55.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- synology-dsm-cinder-iscsi http://openstack-ci-logs.synology.com/73/755373/8/check/synology-dsm-cinder-iscsi/db1bca4 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":8},{"id":"14ba7d6e52dff3851b8ad73b6da2d3198270f08f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-11-11 20:13:44.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/73/755373/8/check/DellEMC_PowerStore_iSCSI/a401149/DellEMC_PowerStore_iSCSI/79 : SUCCESS in 1h 01m 56s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/73/755373/8/check/DellEMC_PowerStore_FC/1da5ab1/DellEMC_PowerStore_FC/75 : SUCCESS in 1h 06m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"22644a581e9212073e8edb03b2cf864f407acb56","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-11-11 20:47:21.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/73/755373/8/check/DellEMC_XtremIO_FC/ab37c8d/DellEMC_XtremIO_FC/65 : SUCCESS in 1h 40m 15s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/73/755373/8/check/DellEMC_XtremIO_iSCSI/9adc5a4/DellEMC_XtremIO_iSCSI/63 : SUCCESS in 1h 36m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"5c89eff44c945c483bf8f052a26c825ae6afc758","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-11 20:49:07.000000000","message":"Patch Set 8:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/73/755373/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4106a00ba207745d942bf840ae169d462b304e12","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-11-11 20:57:57.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/73/755373/8/upstream-solidfire-check/cinder-solidfire-iSCSI/8cb4b23/ : SUCCESS in 1h 49m 38s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"ca5751f4a1d0d4534a9fd55209ced55c355880a1","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-11-11 21:04:17.000000000","message":"Patch Set 8:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/755373-8/Seagate-iSCSI : SUCCESS in 45:06.68\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/755373-8/Seagate-FC : SUCCESS in 38:53.12","accounts_in_message":[],"_revision_number":8},{"id":"42df8afcc1d1bf85a20a8fd47ca26acb93fdd840","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-11-11 21:16:44.000000000","message":"Patch Set 8:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/755373-8/Lenovo-iSCSI : SUCCESS in 48:48.93\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/755373-8/Lenovo-FC : SUCCESS in 46:50.65","accounts_in_message":[],"_revision_number":8},{"id":"038530185c8cf4c4c1fb80e4dde884e4439ec23b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-11 21:32:40.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"60b3b09facdcd26ed925617e8f3ae90155094fed","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-11 21:50:52.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/73/755373/8/check/DellEMC_VNX_iSCSI/68c2220/DellEMC_VNX_iSCSI/72 : SUCCESS in 2h 43m 49s\n- DellEMC_VNX_FC http://publiclogs.emc.com/73/755373/8/check/DellEMC_VNX_FC/15542b7/DellEMC_VNX_FC/73 : SUCCESS in 1h 03m 25s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"c5663ef31fb48c3e14dfbf12ea75057e5fe6d56d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 22:15:57.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3884a72a29e4c499c56cbd732c64bea : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713081030a214becb93b6c75af31e94e : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c1d5be6461d41078a2bf51341765d52 : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/653250cb8fba43a2be2286b5ec539ba9 : SUCCESS in 23m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482060a4d7de4d269195c6a161402f3e : SUCCESS in 15m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/f63fbed50c034a3d88f97824ac53d2ef : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c282838f0d74a2896c597c510c4d150 : SUCCESS in 1h 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f48a5bbd2754e5781c5cbc758256f62 : SUCCESS in 7m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cfa3d3e76ccb470fbd54c680d2415bb1 : SUCCESS in 1h 09m 08s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d13514dd08a447e9a249e29f13e5b13e : SUCCESS in 1h 18m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a833a728c4a4800956a4431e1d9a4bf : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":8},{"id":"287214d00dc315b52b83d7e3499a689087463374","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 22:15:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"d6e56fe0ea899473adf708ff002bc9451267a838","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 22:18:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5043823c5949a39236569ca0376442 : SUCCESS in 1m 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdc51d9ace024db0b521d279b3f7a1ab : SUCCESS in 56s","accounts_in_message":[],"_revision_number":8},{"id":"486571c9e46ba2a8ba0a7725f0456bbc55cd94cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-11 23:11:20.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"dd8bd55088bfd94793fea82c002a339ba5e73a86","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-11-12 07:00:19.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/73/755373/8/check/DellEMC_PowerMAX_FC/fd12196/DellEMC_PowerMAX_FC/101 : SUCCESS in 4h 13m 16s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/73/755373/8/check/DellEMC_PowerMAX_iSCSI/bdcff60/DellEMC_PowerMAX_iSCSI/93 : FAILURE in 5h 05m 41s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"f1d3e44ad6f02cfc9541e1cd8c0e9b0199098df1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-12 11:25:01.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/73/755373/8/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"33c02513c5f91d19e3b3440b503954f78a19662b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-11-16 20:07:49.000000000","message":"Patch Set 8:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpemsa-ci/755373-8/HPMSA-iSCSI : SUCCESS in 52:45.87\n* HPMSA-FC http://os-logs.tristero.net/hpemsa-ci/755373-8/HPMSA-FC : SUCCESS in 37:09.18","accounts_in_message":[],"_revision_number":8},{"id":"7d28d8f95fa1be3daa73de2c05689d30add9141f","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-11-18 13:12:43.000000000","message":"Patch Set 8: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 622244a755b75a52d079f92d8526f6eb5b517106","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8d0c08b4bb9d32f2c02dcc60f1f482331197e546","revisions":{"31ac5e3de8164132d70eb713104ec9df6706ffd1":{"kind":"REWORK","_number":1,"created":"2020-09-30 19:07:37.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/73/755373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/755373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/755373/1"}}},"commit":{"parents":[{"commit":"c8ce118f0a21a4cf622687a6492e8cccb38dff81","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8ce118f0a21a4cf622687a6492e8cccb38dff81"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 15:32:02.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 19:06:14.000000000","tz":-180},"subject":"NetApp SolidFire: Fix API error moving primaries from node","message":"NetApp SolidFire: Fix API error moving primaries from node\n\nWhen SolidFire is under heavy load or being upgraded, the\nSolidFire cluster may automatically move connections from primary\nto secondary nodes, in order to rebalance cluster workload.\n\nAlthough this operation ocurrs very quickly, if a operation is made\nto a volume at the same time it\u0027s being moved, there might be a\nchance that API calls such as create snapshot could fail with\nxNotPrimary error. Normally this will succeed on a retry of the\noperation.\n\nThis patch fix this issue by adding the xNotPrimary exception to\nour list of retryable exceptions in the SolidFire driver.\n\nChange-Id: I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b\nCloses-Bug: #1891914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31ac5e3de8164132d70eb713104ec9df6706ffd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31ac5e3de8164132d70eb713104ec9df6706ffd1"}]},"branch":"refs/heads/master"},"fcd988eb1d370188719b66734e3221c398191a95":{"kind":"REWORK","_number":2,"created":"2020-10-06 11:52:22.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/73/755373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/755373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/755373/2"}}},"commit":{"parents":[{"commit":"c8ce118f0a21a4cf622687a6492e8cccb38dff81","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8ce118f0a21a4cf622687a6492e8cccb38dff81"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 15:32:02.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-06 11:52:08.000000000","tz":0},"subject":"NetApp SolidFire: Fix error moving primaries from node","message":"NetApp SolidFire: Fix error moving primaries from node\n\nWhen SolidFire is under heavy load or being upgraded, the\nSolidFire cluster may automatically move connections from primary\nto secondary nodes, in order to rebalance cluster workload.\n\nAlthough this operation ocurrs very quickly, if a operation is made\nto a volume at the same time it\u0027s being moved, there might be a\nchance that API calls such as create snapshot could fail with\nxNotPrimary error. Normally this will succeed on a retry of the\noperation.\n\nThis patch fix this issue by adding the xNotPrimary exception to\nour list of retryable exceptions in the SolidFire driver.\n\nChange-Id: I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b\nCloses-Bug: #1891914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fcd988eb1d370188719b66734e3221c398191a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fcd988eb1d370188719b66734e3221c398191a95"}]},"branch":"refs/heads/master"},"7068b833f89e32258bc43dc42b0f7bce7155931d":{"kind":"REWORK","_number":3,"created":"2020-10-14 18:31:52.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/73/755373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/755373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/755373/3"}}},"commit":{"parents":[{"commit":"c8ce118f0a21a4cf622687a6492e8cccb38dff81","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8ce118f0a21a4cf622687a6492e8cccb38dff81"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 15:32:02.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-14 18:31:47.000000000","tz":0},"subject":"NetApp SolidFire: Fix error on cluster workload rebalancing","message":"NetApp SolidFire: Fix error on cluster workload rebalancing\n\nWhen SolidFire is under heavy load or being upgraded, the\nSolidFire cluster may automatically move connections from primary\nto secondary nodes, in order to rebalance cluster workload.\n\nAlthough this operation ocurrs very quickly, if an operation is made\nto a volume at the same time it\u0027s being moved, there might be a\nchance that API calls such as create snapshot could fail with\nxNotPrimary error. Normally this will succeed on a retry of the\noperation.\n\nThis patch fixes this issue by adding the xNotPrimary exception to\nour list of retryable exceptions in the SolidFire driver.\n\nChange-Id: I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b\nCloses-Bug: #1891914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7068b833f89e32258bc43dc42b0f7bce7155931d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7068b833f89e32258bc43dc42b0f7bce7155931d"}]},"branch":"refs/heads/master"},"8355df7c4afdfb2acc1f0353067c5dfe36c07ba8":{"kind":"REWORK","_number":4,"created":"2020-10-14 18:33:12.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/73/755373/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/755373/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/755373/4"}}},"commit":{"parents":[{"commit":"c8ce118f0a21a4cf622687a6492e8cccb38dff81","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8ce118f0a21a4cf622687a6492e8cccb38dff81"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 15:32:02.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-14 18:33:09.000000000","tz":0},"subject":"NetApp SolidFire: Fix error on cluster workload rebalancing","message":"NetApp SolidFire: Fix error on cluster workload rebalancing\n\nWhen SolidFire is under heavy load or being upgraded, the\nSolidFire cluster may automatically move connections from primary\nto secondary nodes, in order to rebalance cluster workload.\n\nAlthough this operation ocurrs very quickly, if an operation is made\nto a volume at the same time it\u0027s being moved, there might be a\nchance that API calls such as create snapshot could fail with\nxNotPrimary error. Normally this will succeed on a retry of the\noperation.\n\nThis patch fixes this issue by adding the xNotPrimary exception to\nour list of retryable exceptions in the SolidFire driver.\n\nChange-Id: I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b\nCloses-Bug: #1891914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8355df7c4afdfb2acc1f0353067c5dfe36c07ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8355df7c4afdfb2acc1f0353067c5dfe36c07ba8"}]},"branch":"refs/heads/master"},"208da36a1defc2bcbe83702c987ca6685b5510e5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-14 18:34:10.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/73/755373/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/755373/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/755373/5"}}},"commit":{"parents":[{"commit":"50a6d50c434870be011b4e29a60b4008e0fd6f36","subject":"Merge \"Remove six of dir cinder/scheduler/*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50a6d50c434870be011b4e29a60b4008e0fd6f36"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 15:32:02.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-14 18:34:10.000000000","tz":0},"subject":"NetApp SolidFire: Fix error on cluster workload rebalancing","message":"NetApp SolidFire: Fix error on cluster workload rebalancing\n\nWhen SolidFire is under heavy load or being upgraded, the\nSolidFire cluster may automatically move connections from primary\nto secondary nodes, in order to rebalance cluster workload.\n\nAlthough this operation ocurrs very quickly, if an operation is made\nto a volume at the same time it\u0027s being moved, there might be a\nchance that API calls such as create snapshot could fail with\nxNotPrimary error. Normally this will succeed on a retry of the\noperation.\n\nThis patch fixes this issue by adding the xNotPrimary exception to\nour list of retryable exceptions in the SolidFire driver.\n\nChange-Id: I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b\nCloses-Bug: #1891914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/208da36a1defc2bcbe83702c987ca6685b5510e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/208da36a1defc2bcbe83702c987ca6685b5510e5"}]},"branch":"refs/heads/master"},"f359fbf5a8d5d801bad2a752448a2e42f62a2751":{"kind":"REWORK","_number":6,"created":"2020-10-14 18:42:06.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/73/755373/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/755373/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/755373/6"}}},"commit":{"parents":[{"commit":"50a6d50c434870be011b4e29a60b4008e0fd6f36","subject":"Merge \"Remove six of dir cinder/scheduler/*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50a6d50c434870be011b4e29a60b4008e0fd6f36"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 15:32:02.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-14 18:41:15.000000000","tz":-180},"subject":"NetApp SolidFire: Fix error on cluster workload rebalancing","message":"NetApp SolidFire: Fix error on cluster workload rebalancing\n\nWhen SolidFire is under heavy load or being upgraded, the\nSolidFire cluster may automatically move connections from primary\nto secondary nodes, in order to rebalance cluster workload.\n\nAlthough this operation ocurrs very quickly, if an operation is made\nto a volume at the same time it\u0027s being moved, there might be a\nchance that API calls such as create snapshot could fail with\nxNotPrimary error. Normally this will succeed on a retry of the\noperation.\n\nThis patch fixes this issue by adding the xNotPrimary exception to\nour list of retryable exceptions in the SolidFire driver.\n\nChange-Id: I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b\nCloses-Bug: #1891914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f359fbf5a8d5d801bad2a752448a2e42f62a2751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f359fbf5a8d5d801bad2a752448a2e42f62a2751"}]},"branch":"refs/heads/master"},"f241072c38a2063897f085595565c40789acea6b":{"kind":"REWORK","_number":7,"created":"2020-10-30 14:40:24.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/73/755373/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/755373/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/755373/7"}}},"commit":{"parents":[{"commit":"b91d5474d38a0e1998bbbb4e2208e101f6bf3d22","subject":"Merge \"Fix invalid asert_called_with statement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b91d5474d38a0e1998bbbb4e2208e101f6bf3d22"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 15:32:02.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-30 14:40:07.000000000","tz":-180},"subject":"NetApp SolidFire: Fix error on cluster workload rebalancing","message":"NetApp SolidFire: Fix error on cluster workload rebalancing\n\nWhen SolidFire is under heavy load or being upgraded, the\nSolidFire cluster may automatically move connections from primary\nto secondary nodes, in order to rebalance cluster workload.\n\nAlthough this operation ocurrs very quickly, if an operation is made\nto a volume at the same time it\u0027s being moved, there might be a\nchance that API calls such as create snapshot could fail with\nxNotPrimary error. Normally this will succeed on a retry of the\noperation.\n\nThis patch fixes this issue by adding the xNotPrimary exception to\nour list of retryable exceptions in the SolidFire driver.\n\nChange-Id: I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b\nCloses-Bug: #1891914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f241072c38a2063897f085595565c40789acea6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f241072c38a2063897f085595565c40789acea6b"}]},"branch":"refs/heads/master"},"8d0c08b4bb9d32f2c02dcc60f1f482331197e546":{"kind":"REWORK","_number":8,"created":"2020-11-06 08:06:50.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/73/755373/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/755373/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/755373/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/755373/8"}}},"commit":{"parents":[{"commit":"4f68296d67e81c7f73980cbd6de4a8f912d60ce4","subject":"Merge \"Delete TSM Backup driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f68296d67e81c7f73980cbd6de4a8f912d60ce4"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-09-30 15:32:02.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-11-06 08:06:36.000000000","tz":-180},"subject":"NetApp SolidFire: Fix error on cluster workload rebalancing","message":"NetApp SolidFire: Fix error on cluster workload rebalancing\n\nWhen SolidFire is under heavy load or being upgraded, the\nSolidFire cluster may automatically move connections from primary\nto secondary nodes, in order to rebalance cluster workload.\n\nAlthough this operation ocurrs very quickly, if an operation is made\nto a volume at the same time it\u0027s being moved, there might be a\nchance that API calls such as create snapshot could fail with\nxNotPrimary error. Normally this will succeed on a retry of the\noperation.\n\nThis patch fixes this issue by adding the xNotPrimary exception to\nour list of retryable exceptions in the SolidFire driver.\n\nChange-Id: I67dd2bfba37adcb7cda5f1cd08ff641410ec3f6b\nCloses-Bug: #1891914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d0c08b4bb9d32f2c02dcc60f1f482331197e546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d0c08b4bb9d32f2c02dcc60f1f482331197e546"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
