)]}'
{"id":"openstack%2Fcinder~763195","triplet_id":"openstack%2Fcinder~master~I7bad7b5ebaff3d191c0701a9e72ed4c6334ecaa3","project":"openstack/cinder","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I7bad7b5ebaff3d191c0701a9e72ed4c6334ecaa3","subject":"Update secure RBAC check strings with descriptions","status":"MERGED","created":"2020-11-18 14:41:53.000000000","updated":"2021-03-06 19:27:31.000000000","submitted":"2021-03-05 05:29:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"763195-secure-rbac","meta_rev_id":"1fe945b0ce9c4f4662c4dd2936fba46c7c952b73","_number":763195,"virtual_id_number":763195,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 05:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2021-03-02 22:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-03 13:39:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-03 13:39:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 14:41:53.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-11-18 15:03:52.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"REVIEWER"},{"updated":"2020-11-18 15:04:14.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 15:17:12.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 15:38:00.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 15:40:18.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 17:59:12.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-11-18 19:31:59.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 21:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 23:44:53.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 01:21:24.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 02:08:47.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-19 16:12:49.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 08:04:58.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 11:05:11.000000000","updated_by":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"reviewer":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"state":"REVIEWER"},{"updated":"2021-02-16 16:38:41.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-16 18:46:57.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 19:43:47.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 01:24:51.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-02-17 04:15:10.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-02-17 05:14:20.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-17 05:52:34.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 06:27:27.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 11:40:27.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-02-17 13:41:32.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-02-17 15:30:38.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-02-17 21:18:43.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-02-18 11:38:41.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 03:11:13.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-03-02 22:33:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-03 13:39:23.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-04 10:42:41.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-04 23:47:22.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"c860f10630468e05498e932cad9a54d115dd4b4c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-18 14:41:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a091812c9ac993ff382aece6e85799e7daeada60","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-11-18 14:51:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/95/763195/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/665d8d6/665d8d6.html : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c24a8960c87f8d0365cb5c3cbee711bbefc2a792","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-18 14:55:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c08d08ee1ded85ebb90eb133568c1e1193fa9230","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-18 15:02:16.000000000","message":"Topic changed from 763195 to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"0502906dd35b0ab6fc321cf1fa7281739b38b981","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-11-18 15:03:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/763195/2/iscsi/ : FAILURE in 25s\n- cinder-smb http://cloudbase-ci.com/cinder/763195/2/smb/ : FAILURE in 7m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a1789299c31f2ad890b3b3993bd22ef0cb6b2cf4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-11-18 15:04:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/95/763195/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a403024/a403024.html : POST_FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"acdf8d6f97fed32246b65988628df836ec597db5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-18 15:17:12.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/95/763195/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/70d0fc5 : FAILURE in 21m 14s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/95/763195/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/4013827 : FAILURE in 20m 43s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"1653e5cea7c0cfe62de503a28fbf73716d57df12","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-11-18 15:23:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/8a534eb51f1f4bf27a3dcf92d1431d95cbb044a3/index.html : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"20c4dd741d9ca8b88972f01d4b355b8af9af5cca","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-18 15:38:00.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/95/763195/2/check-not-vote/ext-cinder-zuul/71c7ad0 : SUCCESS in 41m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3b981cf0999143ebd16d10848f635c5e327302f0","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-11-18 15:40:18.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/b3e4b7608ca674cc83730ded332fddd2755d7af0/index.html : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"58e342987c477f76d2d2245f2785cce89ff7ea1c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-18 17:59:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/2/check/dsvm-tempest-cisco-zonemanager-job/53a4faa/ : FAILURE in 1h 30m 21s","accounts_in_message":[],"_revision_number":2},{"id":"1fa0e1a05ae85338a95a6bf7f92ce5f11de05ebe","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-18 18:25:27.000000000","message":"Patch Set 1:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/763195/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cd05c756db071002c6546f03d2909f5e85d61a32","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-18 19:02:44.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/763195/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f36e0c0ea2ea59afcdcbeafc9b90ed8166076243","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-11-18 19:31:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/6735f2157004442685fc335c3a25cc16 : SUCCESS in 1h 10m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2b7e7fa71d15fb2372f81d8c190f1565a3f164db","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-18 20:19:52.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/763195/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1ebacb49bf5bb678eb2431a751edd16cee7c3d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 21:29:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfcd4d017fee416c8c3efed3b74462e9 : SUCCESS in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33f8e23bbda46d59e8f38291fcda559 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a1e84887d294a388f9775c298677880 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/610ccad0c9104ddbaa5d5acd7d043439 : SUCCESS in 11m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c074dfbc29034ed1ba07f30eadda5421 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05924051e6d14923861c6c119d1b341d : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ab694ad3011b4b1da917dcccbce4b852 : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a2c691b70f9442788ccec9d56e51223 : SUCCESS in 1h 42m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5790faf5aff5415fa129569f79a19b74 : SUCCESS in 12m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/756c31bcde97464e999331a1f6d1358e : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d728af584a84eeebed91c141db319ed : SUCCESS in 5m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6209044c7b2b417dae33edf719e9aa3f : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/48d24ef0ea6147ea83d598ca749cea7c : SUCCESS in 44m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2dc230c26b1e4e108d231ab556ee7da9 : FAILURE in 4m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/139603f2be5f4d139655f37265aa5ce1 : SUCCESS in 1h 11m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0baa2e5690c5477fb18baec9a88eb801 : FAILURE in 1h 19m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4dc818b504534540ac8d6e789165f728 : SUCCESS in 1h 33m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c77eb70447774e91878764eb2b247fa2 : FAILURE in 1h 12m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29256dc7410047dcb95e687222eac735 : FAILURE in 1h 27m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f91ddea4bb1c4e9aac6134f22ef9de1a : SUCCESS in 1h 24m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26e6c7a6894c4eedb2d70b16396a40c0 : SUCCESS in 2h 08m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fe34261cf524e0181b201f4d97d2c1c : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":2},{"id":"00eb15adc6c5eaa6a275526b8439082e5537454f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-18 23:44:53.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/57be1aaec56344dd83b45c8d1d9c37d4 : RETRY_LIMIT in 24s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/763195/2/check/devstack-plugin-ceph-dsvm-focal-py3/4fd70eb/ : SUCCESS in 1h 39m 15s","accounts_in_message":[],"_revision_number":2},{"id":"6ecfb3d51d37b0edbb98e80aaaaf7bf332e673d5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-19 00:09:17.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-1/ns5_nfs : SUCCESS in 7h 27m 55s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-1/ns5_iscsi : SUCCESS in 7h 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"76a39928400096595f32d0ae5130f9af7ff46fc8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-19 01:21:24.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-2/ns5_nfs : SUCCESS in 8h 35m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-2/ns5_iscsi : SUCCESS in 7h 16m 55s","accounts_in_message":[],"_revision_number":2},{"id":"0287acab9655473a14f53342294480e131ebfc67","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-19 02:08:47.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/763195/2/check/DellEMC_Unity_FC/bef30fd/DellEMC_Unity_FC/130 : FAILURE in 3s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/763195/2/check/DellEMC_Unity_iSCSI/f73ad12/DellEMC_Unity_iSCSI/115 : SUCCESS in 1h 52m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3595e4e27365822d9dd40913813b72688be3807","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-11-19 16:12:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/95/763195/2/20201119220438/NEC-ISCSI : SUCCESS in 3h 00m 35s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/95/763195/2/20201119220438/NEC-FC : SUCCESS in 3h 08m 08s","accounts_in_message":[],"_revision_number":2},{"id":"65ca91eda38c868445ffe953e01e519c4bd5cb08","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-19 17:05:11.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/763195/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"72e37b393f011c0473d78e63809b9b4a00d8a15a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-11-20 08:04:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/95/763195/2/check/DellEMC_PowerMAX_FC/3320183/DellEMC_PowerMAX_FC/140 : SUCCESS in 3h 57m 44s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/95/763195/2/check/DellEMC_PowerMAX_iSCSI/463a781/DellEMC_PowerMAX_iSCSI/131 : SUCCESS in 4h 55m 28s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"951a23ef8ae6bf2fca94a71e090d15967db2c5eb","author":{"_account_id":24921,"name":"Hedvig CI","email":"openstack@hedviginc.com","username":"hedvig_ci"},"date":"2020-11-20 11:05:11.000000000","message":"Patch Set 2:\n\n* Hedvig-CI http://os-logserver.hedviginc.com/refs/changes/95/763195/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"723254ebc67ae459befe9ddd02db7858099a6d81","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-22 01:40:35.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-2/ns5_nfs : SUCCESS in 49h 7m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-2/ns5_iscsi : SUCCESS in 48h 53m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2d07523b7af1b0a017e00fa18db58cd4706d83a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-16 16:02:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"462e42b86377fbd130c71d3b12b32c8378b46076","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-16 16:14:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/95/763195/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d2fd95b/d2fd95b.html : FAILURE in 10m 27s","accounts_in_message":[],"_revision_number":3},{"id":"5ef9afa15f30d98086708567ef38f5bec91ebfbf","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-16 16:38:41.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/95/763195/3/check/dsvm-tempest-TOYOU-ACS5000-driver/444ec7a : FAILURE in 23m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a23620313d63324b0b0a164ce4b05543fce99cd2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-16 17:07:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/95/763195/3/check/iscsi-driver-primera/f37f4fc/f37f4fc.html : FAILURE in 1h 03m 13s","accounts_in_message":[],"_revision_number":3},{"id":"52d477bb1421f625a8e91447de695a1fda27e1bf","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-16 17:11:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/763195/3/iscsi/ : SUCCESS in 1h 08m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/763195/3/smb/ : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":3},{"id":"1d850a0b184d38d878a717ca0e3092d680b29487","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-16 17:38:29.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/763195/3/check/tempest-dsvm-full-focal-py3/e4f10e2/ : FAILURE in 34m 08s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/763195/3/check/devstack-plugin-ceph-dsvm-focal-py3/b9f008a/ : FAILURE in 35m 03s","accounts_in_message":[],"_revision_number":3},{"id":"dbf3b5d1cc56731a6e52d4c0d0b9c30daadc2dae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-16 18:46:57.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"74922e91e9dc15d8e638a1574e93f38f7ac24ef1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-16 19:14:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/3/check/dsvm-tempest-cisco-zonemanager-job/e3ea54c/ : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":3},{"id":"dde82ffdc4cf94dc624810ab63472e7ec6b48920","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-16 19:43:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/95/763195/3/check/dsvm-tempest-zadara-driver/e39bad5 : SUCCESS in 1h 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f754801ad5bfc9ce0b0a38b7faf8e6e13b01a0d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-16 21:56:57.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-3/ns5_nfs : FAILURE in 5h 14s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-3/ns5_iscsi : SUCCESS in 4h 27m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"9db993722d9fa9cadd4f4589ff6f2d05874733c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-16 22:15:18.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"10d35012971de36fc991ddebd7a811f4956e4b54","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-16 23:57:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/95/763195/3/ : SUCCESS in 1h 12m 41s","accounts_in_message":[],"_revision_number":3},{"id":"f32d512c126afc9644418e451ac8fc1896dda3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 00:30:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8871f0ba1aa64113b293a58d50cad2f0 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffcb750f66e4945a4db36b5fd0ba517 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9885cf2e7c8a4935b5769c5d75abf5a8 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56e5f84477a04817b584402ad766fd4c : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14f87c217c17479cbff596edc2cbf942 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae7a2c200db4e36aca407a384a76681 : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/15a75e324bab4c119e723c180343f9b8 : SUCCESS in 1h 01m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2bc76015bd743ea8925caab5b10e2ca : SUCCESS in 1h 14m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6092a478d3fe4d2bb67d96eed639ccb3 : SUCCESS in 10m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef6b264230234902979cf61c16720f8f : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89608a813fbd4cbe8c39debacd44a244 : SUCCESS in 7m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5424b6027cba4b42b637b048deb907a4 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8003b69be4c44daba211194fb090fb65 : SUCCESS in 40m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f879b2801e5045e4a69976c9b6b68a1d : SUCCESS in 5m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6bcf7167a02443948545b98236a725e3 : SUCCESS in 1h 09m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ac088de22b9d471e9dcbda8c965be945 : FAILURE in 1h 23m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf6cef071cd94d1785fd384fc31c557d : SUCCESS in 1h 44m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a079d9d748ec4a1cb9bac2e57be3025a : SUCCESS in 1h 17m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/79053b23009042bbb75e63895c49a9cf : SUCCESS in 1h 21m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3f2de5e9c93b4d8a8de696e701974237 : SUCCESS in 1h 25m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c86d31ec378a45ae892eeae67807e5bb : SUCCESS in 2h 25m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70c666ef742d4fcda0327f9dc3e79b96 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":3},{"id":"0ecdda9ec28c2f5be3edb7937b708b6938d5f9db","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-17 01:24:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/95/763195/3/upstream-solidfire-check/cinder-solidfire-iSCSI/bcaf07c/ : SUCCESS in 53m 11s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"4afafec765d5f7da931d9146798b1bba3908aedc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-17 04:15:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/95/763195/3/check/DellEMC_PowerStore_iSCSI/6b4d972/DellEMC_PowerStore_iSCSI/415 : FAILURE in 59m 43s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/95/763195/3/check/DellEMC_PowerStore_FC/7657f3b/DellEMC_PowerStore_FC/409 : SUCCESS in 50m 59s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"792db75c1670de2e1c8ed471de3d5fcf0357a738","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-17 05:14:20.000000000","message":"Patch Set 3:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/763195-3/PVME-iSCSI : FAILURE in 15:18.86\n* PVME-FC http://os-logs.tristero.net/dell-ci/763195-3/PVME-FC : FAILURE in 15:26.84","accounts_in_message":[],"_revision_number":3},{"id":"e0b3d636a9faf47571a5d52086eb617308fc050f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-17 05:52:34.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/95/763195/3/upstream-check/cinder-cDOT-iSCSI/3ba62e0/ : SUCCESS in 37m 16s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/95/763195/3/upstream-check/cinder-cDOT-iSCSI-IPv6/0461723/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/95/763195/3/upstream-check/cinder-cDOT-NFS/63abc8f/ : FAILURE in 2h 53m 31s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/95/763195/3/upstream-check/cinder-cDOT-NFS-IPv6/7d6820c/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/95/763195/3/upstream-check/cinder-cDOT-FCP/36256b8/ : FAILURE in 3h 06m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"484a1958daef7c4802fdee1b07ebfccdfa739313","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-17 06:27:27.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/763195/3/check/DellEMC_XtremIO_FC/97f4a76/DellEMC_XtremIO_FC/771 : FAILURE in 2h 01m 56s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/763195/3/check/DellEMC_XtremIO_iSCSI/305639f/DellEMC_XtremIO_iSCSI/773 : SUCCESS in 1h 40m 53s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"7ff482390b2a30c7cb272b64944ffea5208d1db7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-17 06:32:12.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"15c45d880f96545a064048cea0171bc0ec95f21b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-17 06:58:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/763195/3/check/DellEMC_Unity_FC/85678c9/DellEMC_Unity_FC/710 : FAILURE in 2h 00m 13s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/763195/3/check/DellEMC_Unity_iSCSI/e860bbb/DellEMC_Unity_iSCSI/671 : SUCCESS in 2h 14m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"9baa5fe0e851e6598aa9bdc08ec67090b30111fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-17 07:06:38.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"6e308cd89076d800e340daf3a1440d08a157ce0f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-17 10:39:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/95/763195/3/20210217170706/NEC-ISCSI : SUCCESS in 2h 17m 25s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/95/763195/3/20210217170706/NEC-FC : SUCCESS in 2h 32m 16s","accounts_in_message":[],"_revision_number":3},{"id":"39b42867c0be522120b46c6683faa37e97dc9b99","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-17 11:40:27.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/763195-3/Lenovo-iSCSI : SUCCESS in 48:54.75\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/763195-3/Lenovo-FC : SUCCESS in 42:45.27","accounts_in_message":[],"_revision_number":3},{"id":"325599919e84eff0a3bfe4bfa8194c68c2a13c3e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-17 13:41:32.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/95/763195/3/check/DellEMC_SC_FC/2d3b5a7/DellEMC_SC_FC/342 : FAILURE in 1h 58m 02s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/95/763195/3/check/DellEMC_SC_iSCSI/3e67d07/DellEMC_SC_iSCSI/341 : FAILURE in 2h 14m 19s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"50c00d39945ec4d77860d48a8db47be8172c1074","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-17 15:30:38.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/763195-3/HPMSA-iSCSI : FAILURE in 1:06:57\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/763195-3/HPMSA-FC : SUCCESS in 47:33.82","accounts_in_message":[],"_revision_number":3},{"id":"1274e4d7afff4d9bf81ebde1277529e03f29f14c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-17 17:48:52.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"21df8ae1650df6d2951b6feccc7788194882828f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 18:30:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/95/763195/4/check/iscsi-driver-primera/a4d0503/a4d0503.html : FAILURE in 8m 57s","accounts_in_message":[],"_revision_number":4},{"id":"44e36599a3797435d89729ac2632974e306f8032","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 18:34:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/95/763195/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/770d9c9/770d9c9.html : FAILURE in 13m 23s","accounts_in_message":[],"_revision_number":4},{"id":"ca65c35ae1bf78a5b2e559f3674a279e86b12d6f","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-17 18:51:54.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/95/763195/4/check/dsvm-tempest-TOYOU-ACS5000-driver/c8613f5 : FAILURE in 23m 20s","accounts_in_message":[],"_revision_number":4},{"id":"c8da3b2d124dc0b6d29b09452729efb8be214e14","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-17 19:07:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/95/763195/4/check/dsvm-tempest-zadara-driver/434cd50 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 17m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5e345b279321f49f4daa27fa84830ed849ffbd3a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-17 19:42:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/4/check/dsvm-tempest-cisco-zonemanager-job/a770d69/ : SUCCESS in 1h 18m 44s","accounts_in_message":[],"_revision_number":4},{"id":"77ee4a865e514c28afe02ebff2b48ddfd4ca2954","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-17 19:52:10.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/763195/4/iscsi/ : FAILURE in 1h 38m 46s\n- cinder-smb http://cloudbase-ci.com/cinder/763195/4/smb/ : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":4},{"id":"ed6a5b5ecd23f6d326085f0e5834c693e5c72f9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 20:59:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7cd52ad9af147c9a3cee1591eb925e9 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d31c87c5dc84b4ca0efae4a733c5079 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19d7ae7901df42adac4b05e58bb91d99 : SUCCESS in 13m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a03604a61e52488e9bfc597754725cd9 : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2dcf174aa8b408fb624a6b17aedda58 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49337aa668514ce0b8ac2de31d51868a : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/64dfb156c78a4c68b7f6a572ec1fc460 : SUCCESS in 1h 02m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/943edbcc45fb42c08a2d5edad3b7c775 : SUCCESS in 1h 09m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4a25631657bc4e178efa1d1cf3ba0a83 : SUCCESS in 12m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4c1be2e431504e70aa41e96812665ae5 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4335c3d99ef94dd3a7e3575d4fd1fe90 : SUCCESS in 6m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b77cd6f42dff4faca1209ae4495ee854 : SUCCESS in 8m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/88c078891f4e442f96593034edb59b89 : SUCCESS in 40m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7db1a19c366144fca62a383a6486f02f : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/254f9e67315c44c499504286d1a3e148 : SUCCESS in 1h 15m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cdda2ad157164226a2c9543453df8aa7 : TIMED_OUT in 2h 06m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf4f9d7fe44e4aa3b7f198a8048c3ea4 : SUCCESS in 1h 38m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/44d500b04f3b4ea787ace5256dae3684 : SUCCESS in 1h 07m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c41eb629323a449c80e96df9c14c501c : SUCCESS in 2h 08m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d401cd9e3fef40388c83ae9b9487c514 : SUCCESS in 1h 24m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/339f8077e97048ca8ed6c6f0c7a01250 : SUCCESS in 2h 19m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0638e152c00f4ad1996f5fe7138ce482 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":4},{"id":"955fa84c27e59fb38cd3ae2d33c0c0b64d302ec4","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-17 21:18:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-95-763195-4/1613595568 : SUCCESS in 19m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"df5a641c39d3f5e57e50648edc47ae7a3705b943","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-17 21:42:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-iSCSI/2d4b003/ : SUCCESS in 41m 20s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-iSCSI-IPv6/d5de248/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-NFS/9d60b90/ : SUCCESS in 39m 14s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-NFS-IPv6/30e8b4e/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-FCP/8d2ee3d/ : FAILURE in 9m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"df73a051b03673687e086f6e4d4fc4916bf84b15","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-17 21:54:37.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerStore_iSCSI/c9b7be3/DellEMC_PowerStore_iSCSI/431 : SUCCESS in 54m 39s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerStore_FC/85427a2/DellEMC_PowerStore_FC/425 : SUCCESS in 47m 11s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"120cd7361396232c03e21b618eb4fac8634850dd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-17 22:25:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/95/763195/4/upstream-solidfire-check/cinder-solidfire-iSCSI/2ee4c7d/ : SUCCESS in 1h 24m 05s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"e9bd682a7f1c7cd513f16ad8e068408a7273e266","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-17 22:44:22.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/763195/4/check/tempest-dsvm-full-focal-py3/689ce2d/ : SUCCESS in 1h 59m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/95/763195/4/check/devstack-plugin-ceph-dsvm-focal-py3/7932eab/ : SUCCESS in 2h 03m 13s","accounts_in_message":[],"_revision_number":4},{"id":"3376d1d36cb5685c7bd3dfec1ac891ee4d04f11c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-18 00:04:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/95/763195/4/ : SUCCESS in 1h 15m 11s","accounts_in_message":[],"_revision_number":4},{"id":"27942ad27f1048a2eca83d9bff98b4bfa4028fec","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-18 07:14:11.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-4/ns5_nfs : SUCCESS in 5h 48m 7s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-4/ns5_iscsi : SUCCESS in 5h 38m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"49cf350d11706dac1f5263ce095fa106cda83905","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 11:38:41.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_VNX_iSCSI/85bd44a/DellEMC_VNX_iSCSI/401 : FAILURE in 1h 20m 14s\n- DellEMC_VNX_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_VNX_FC/70ca798/DellEMC_VNX_FC/420 : FAILURE in 1h 00m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"3213ec9410f03a6ef59eabed6b32e97324ae17e6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 13:09:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_XtremIO_FC/8cfec7d/DellEMC_XtremIO_FC/788 : FAILURE in 2h 04m 59s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_XtremIO_iSCSI/8c45ac9/DellEMC_XtremIO_iSCSI/790 : SUCCESS in 1h 41m 43s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"be9a8cb4148a123f6f9d46c2d578280187817fbe","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-18 13:24:07.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/763195-4/PVME-iSCSI : SUCCESS in 49:43.68\n* PVME-FC http://os-logs.tristero.net/dell-ci/763195-4/PVME-FC : SUCCESS in 47:06.92","accounts_in_message":[],"_revision_number":4},{"id":"17185a190244052e7c6d2d1f02ce320e6e197bc3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-18 15:36:44.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_Unity_FC/c9eb80f/DellEMC_Unity_FC/726 : FAILURE in 2h 02m 11s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_Unity_iSCSI/bbedebd/DellEMC_Unity_iSCSI/688 : FAILURE in 1h 56m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"5b26ab682d2dfeed45c3d4f8f567493c9b13d467","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-18 18:11:22.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"25c34b966793e76032472a054c63beae2f45e0a3","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-02-19 03:11:13.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/763195-4/Seagate-iSCSI : SUCCESS in 42:05.73\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/763195-4/Seagate-FC : SUCCESS in 39:42.58","accounts_in_message":[],"_revision_number":4},{"id":"cbf3e8f19dddcceb845688ec8276577875a6922c","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-19 06:24:44.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/763195-4/Lenovo-iSCSI : SUCCESS in 45:19.57\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/763195-4/Lenovo-FC : SUCCESS in 42:21.61","accounts_in_message":[],"_revision_number":4},{"id":"ea03ef5dbe7df69eb43286ed7d78ecd1d9217196","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-20 05:38:36.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-4/ns5_nfs : FAILURE in 6h 55m 21s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-4/ns5_iscsi : FAILURE in 7h 1m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"5d339b33b63aaabb64baf7384b4307c58ad12e9c","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-20 06:27:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/95/763195/4/20210220130905/NEC-FC : SUCCESS in 2h 16m 08s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/95/763195/4/20210220130905/NEC-ISCSI : SUCCESS in 2h 18m 49s","accounts_in_message":[],"_revision_number":4},{"id":"d371e365f2bacd0ea2299f28dffec543b2d23d0e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-21 15:54:09.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerMAX_FC/75d65e3/DellEMC_PowerMAX_FC/483 : SUCCESS in 6h 02m 43s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerMAX_iSCSI/5f44a18/DellEMC_PowerMAX_iSCSI/447 : SUCCESS in 4h 53m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"39bcd4e17238340aa444ad4b06bfdbb3c9402d4c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-21 17:14:47.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"cd5efbca918b20cff8c1757fc54b1631222f5f40","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-22 15:14:22.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-4/ns5_nfs : FAILURE in 4h 9m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-4/ns5_iscsi : FAILURE in 4h 23m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"1297ce0433ac40a1553864dd42a27931e8350132","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-22 17:21:00.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"dae0488cca1149905ff49c4534a055d803184e35","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-23 21:21:38.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-4/ns5_nfs : SUCCESS in 4h 10m 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-763195-4/ns5_iscsi : SUCCESS in 3h 57m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"50922e25063ffa8bee4f01098dc4b5144318d1a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-24 00:27:19.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"017fd7da32e50a752cbe4f0d2f2ee0ecfe8029cd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-02 22:33:47.000000000","message":"Patch Set 4: Code-Review+2\n\nThe added comments look clear to me.","accounts_in_message":[],"_revision_number":4},{"id":"0288898ef35e8b7ca06da0d3eaf9d985d9d577b5","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-03 13:39:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThis looks good.  There is one small comment but it can be handled with a follow-up patch.","accounts_in_message":[],"_revision_number":4},{"id":"fcbed4415f6597bbb25dc2b19184439cdfe79260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 13:40:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"da1d1e5854de738e69c7994dd859f675ba446018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 18:17:54.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae78623ada2c4d28bf73bb8cf70f0406 : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0872b55e59f7420f9518b566436e9c38 : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62f682451e404865b3d066b8def94216 : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9034b9efecad41a2a4cb52ffe3c93285 : SUCCESS in 14m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c56c11e4a6e431ca6dd0c086b83cf5b : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/29037fa935704287bc9c4f6946437a6e : SUCCESS in 1h 06m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f455b88ac72488db08e4114c3b8dd8b : FAILURE in 1h 18m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/72b25c7b79c1476d8264171af09f1980 : SUCCESS in 1h 25m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f04fb7adbe104914ac29bc9fd5871a25 : SUCCESS in 1h 32m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb847857c4c74a84b3cca3c69a9e7d73 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":4},{"id":"2acdbfea72a42d7348d1f7a2e7b1c1d0a92bd7fc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-03 22:57:52.000000000","message":"Patch Set 4:\n\nrecheck tempest-integrated-storage couldn\u0027t delete a volume during tear down","accounts_in_message":[],"_revision_number":4},{"id":"0230312cba7ea2ca7b6fcac9a29eef4d738c7fb9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-03 22:58:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62c996bb997c2ddf734beeeda5ef583fba2ed807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:18:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e35df14bfe9444bead29d04b2964e542 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77fbf437aed64e9a8eca66c5dfefec87 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d8dfeca5b140d39de37a30546bf0af : SUCCESS in 11m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1feb6335e24343a1bbd9cd867579bf89 : SUCCESS in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/411f81272c214576b1c441f5b87e3d3e : SUCCESS in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5215c458c8094bec994b280e05c56b7e : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/80721d821d7c491ba0c3f9f9bae05b58 : SUCCESS in 1h 14m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ed968a8e323470e840c9eaf6af8e3ff : SUCCESS in 1h 04m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/99959355778a41af8ff7d1e112e23b00 : SUCCESS in 13m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f9474c59600149e9a8b13ffab7ecf260 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a3b25f2c58c47e99d180927e9aca1a6 : SUCCESS in 9m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8b65aa9b9f3342eaae7357158b208ecc : SUCCESS in 9m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/84a86ed429d447d0b4dbf352f368e243 : SUCCESS in 45m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/df036c8d9be0479a90f605744d436815 : SUCCESS in 7m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0be692bcd5084987a957e219eb603eb2 : SUCCESS in 1h 13m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a156dc6091b04e24ad18b50b9ed156b4 : FAILURE in 1h 40m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9cb846eb24da4ad092073c0de44ce75e : SUCCESS in 2h 01m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b041eeebdf7b4d9c9ab6d7c3ed47fdd0 : SUCCESS in 1h 08m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/500f0a1f262f493386d2a27107983e5d : SUCCESS in 1h 14m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8d64047c397841d4a6d4cb469f9b2605 : SUCCESS in 1h 18m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d18cfa1d522430fa580b7da39062eb5 : SUCCESS in 2h 07m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79c6a24b739e4b318c666f6a227702c9 : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":4},{"id":"cc0e24be842deb6a995d98cc6f745dee50452e23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:19:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a87b9dc5720a1e721eee0692810afc54cd263bf4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-04 03:10:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/95/763195/4/upstream-solidfire-check/cinder-solidfire-iSCSI/989b091/ : FAILURE in 56m 40s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"d38653df3ac095b3cd3a09ae94099dc6042ae558","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 03:19:07.000000000","message":"Patch Set 4:\n\nthe openstack-tox-py36 failure seems to happen intermittently.  The _notify function is mocked in that test, so there shouldn\u0027t be any interaction with other tests.","accounts_in_message":[],"_revision_number":4},{"id":"944a668c50abcf29490d97d940db4c06e4fd6f20","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 04:10:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/4/check/dsvm-tempest-cisco-zonemanager-job/e82fb5f/ : SUCCESS in 1h 19m 38s","accounts_in_message":[],"_revision_number":4},{"id":"eefb5dc1ab7da0d4a9e95683cc5e440f6a8debc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 04:42:37.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffec959bb07b422998d8e07e3d1139e8 : SUCCESS in 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63c8a9a9520417b82476acb50c90105 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4f5a48da5b24bb1a01fe5176ea01692 : FAILURE in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7b579a755a646ac885d18b1fb9b444d : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe2f174f1174776986ca372b5ffbe7d : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/01efa65f0f0d4e779e9b4c82d0963cec : SUCCESS in 47m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/535804bb025f4d179c4c87e282653b60 : SUCCESS in 1h 22m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/efc35a748d1d432f83c3a55e3a558a26 : SUCCESS in 1h 08m 42s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/34069576bb4c4df8a23c0090702eca85 : SUCCESS in 1h 07m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61512b58bc964bb4982dc54c8a9df749 : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":4},{"id":"c0a582e531959320fb60c89e2be4f43fb8b747bd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-04 05:06:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-iSCSI/f8b2d61/ : SUCCESS in 39m 40s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-iSCSI-IPv6/31b321a/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-NFS/6173356/ : SUCCESS in 38m 01s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-NFS-IPv6/5e4a2e9/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-FCP/65a13f1/ : FAILURE in 3h 38m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"d9c1473af0b8c166bc776c0586416bb21e515142","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 05:06:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6e3f8a667a1aea63a42eb11a06e6d53d139a562c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-04 05:07:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f05ee315c2f2bcedd723703670d935b9a08d99b7","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-04 06:14:17.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/763195-4/PVME-iSCSI : FAILURE in 15:14.17\n* PVME-FC http://os-logs.tristero.net/dell-ci/763195-4/PVME-FC : FAILURE in 15:15.66","accounts_in_message":[],"_revision_number":4},{"id":"ebda838bde3c7d34c1e1645b6af5e37dd3cbd566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 07:19:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c10be35d82e4a4296219dd2e2c78c34 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad67570ee0d4b93b0cc8301ab76c603 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c751d90930904cf7991ad0e09f7daf52 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6fa051f8cbc472b9edf410950403e6e : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/545b9579ffc849ae8fd0661ad6eed80c : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ffe2cc7d82419c92274a0a8c3dd608 : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/107ddd05f0c8444f8ff85d049167d78e : SUCCESS in 58m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f4241e9c9404cb3874a8a9e5b963f2f : SUCCESS in 1h 18m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/38ea822b7df548e0a9a7eea808ee0277 : SUCCESS in 12m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d9cc80800ae465caa4308363c0b1c38 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef4ab600a48343a694edc12d2978473f : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c18f071792a47abb3af755d23b7149e : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b08873b17f3746dbbec83cfb21f859f4 : SUCCESS in 38m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/365c16a047db47908a761266b5cb148e : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b7667eea12a4bef9314f4d4731bb30c : SUCCESS in 1h 19m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/968ec77286ce498f88a79d886a1f22e9 : FAILURE in 1h 32m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1eb9b4b065140b7aa3eb9cd130769aa : FAILURE in 1h 50m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/67f400bdf9a748ab8da83e284b6357c2 : SUCCESS in 1h 14m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dae80af337c1471d82808cc7408d5e3b : SUCCESS in 1h 26m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6d8a39760bb24705b361616ceae8305d : SUCCESS in 1h 28m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51d269f4ed1044a8b71bdebdfbd5ae80 : SUCCESS in 2h 10m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38aad974f5784069aa8d2a67584e3190 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":4},{"id":"1ae61257b9604ba2c605a2af3fe399fffc06e624","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-04 07:40:16.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"eec812725df4266aec1d7abc65855d24440b5c88","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-04 08:35:23.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerStore_iSCSI/0910ac1/DellEMC_PowerStore_iSCSI/502 : SUCCESS in 59m 08s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerStore_FC/4d7e5b8/DellEMC_PowerStore_FC/496 : SUCCESS in 53m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"2ecbcf45548f07fa20298910531e965605be8b95","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-04 10:42:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/95/763195/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1ebd321/ : SUCCESS in 1h 49m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/95/763195/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8dc7fbe/ : SUCCESS in 1h 52m 34s","accounts_in_message":[],"_revision_number":4},{"id":"6e7d390f1f28aa8660324962813a4864b18eb0c9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 11:41:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/4/check/dsvm-tempest-cisco-zonemanager-job/53f33d1/ : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":4},{"id":"c215bfeb60714ec22f068b540d62661dcd76834b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 11:49:41.000000000","message":"Patch Set 4:\n\nrecheck cinder-tempest-plugin-lvm-lio-barbican tempest.api.compute.servers.test_server_actions.ServerActionsTestJSON.test_create_backup had a read timeout","accounts_in_message":[],"_revision_number":4},{"id":"feb4e04e776096d5ea9e29aa4f91aa22257b5401","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-04 11:50:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"952aebbadf9593d0433be16125fcd0cd6afc3fc3","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-04 13:03:15.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/763195-4/Lenovo-iSCSI : SUCCESS in 48:38.01\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/763195-4/Lenovo-FC : SUCCESS in 45:38.37","accounts_in_message":[],"_revision_number":4},{"id":"d1eff4372ae055db947fb66e86c2ac991763ed8c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-04 13:20:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e2a5947153ded60c27934744fc1d44b4c49f2ef8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-04 13:20:25.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1fb14b70963954f339c04e57d441583cd3fd4d0a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-04 13:20:47.000000000","message":"Patch Set 4:\n\nOh - sorry, Brian. I didn\u0027t realize you already issued a recheck.","accounts_in_message":[],"_revision_number":4},{"id":"45d1c4111a11a860f3425b939a32aafd4a2a3210","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 13:42:53.000000000","message":"Patch Set 4:\n\n@Lance: I thought you were trying to issue a proactive recheck because unfortunately the cinder-tempest-plugin-lvm-lio-barbican is currently failing.  This time it is cinder_tempest_plugin.api.volume.test_volume_backup.VolumesBackupsTest.test_incr_backup_create_and_restore_to_an_existing_volume[id-b5d837b0-7066-455d-88fc-4a721a899306] : tempest.exceptions.VolumeRestoreErrorException: Volume e68236e1-3ca4-4e2f-a5e8-5750a6e8197a failed to restore and is in ERROR status","accounts_in_message":[],"_revision_number":4},{"id":"e380c5419e66d9dc8b889a81b9f3ff5bfc3f9afb","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-04 13:54:33.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/763195-4/HPMSA-iSCSI : FAILURE in 1:05:40\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/763195-4/HPMSA-FC : SUCCESS in 39:32.22","accounts_in_message":[],"_revision_number":4},{"id":"45de358899198e980663810f34fc92839c7127d0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 14:01:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/4/check/dsvm-tempest-cisco-zonemanager-job/29aad59/ : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":4},{"id":"d220efb29a8c0f395252b70614221b82594ac00a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:09:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9755a3db0534ba08dbd3d5c005b1b65 : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6adf8bc2e88f41c38bcfcec7a38cec34 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27c7bb251b6d426c9217006dafd24f4c : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5a53dcce017455abf0546c377160669 : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81bb9498590444478e164e12aaeea4e0 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3305c2b86917463bb4f2874b60594722 : SUCCESS in 14m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/796a823222ba4b519a9b7014bdf7d08f : SUCCESS in 1h 08m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9970d229b86e4985a27538e505f12bc7 : SUCCESS in 1h 29m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e312e3d98b4b4558beece8589dc078ef : SUCCESS in 17m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c4a3da59a3da4d708e26ae8fbaa37016 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/37a26080945748d5ab1a45af0c0af7aa : SUCCESS in 7m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9ce0f17e2e2848608ccf096a9a9d16b7 : SUCCESS in 9m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b6ea9e5f6b344f80949277d95843a005 : SUCCESS in 46m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a350d3108ff45248493b57ef59c2e0a : SUCCESS in 10m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c642da1b3a2e4b82b59c5ab387ba2be0 : SUCCESS in 1h 24m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eec7b53f4b6d4c0aa2209653bd564bce : FAILURE in 1h 29m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8978f931835a423387fce858c0cb3ecc : FAILURE in 1h 22m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4edbd3b904644542b9d680f7302fe4b9 : SUCCESS in 1h 23m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3682111c106046ef826a0cf4dd45965f : SUCCESS in 1h 30m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d36e766212524e01ab3e4b52570b4c85 : SUCCESS in 1h 17m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/01fcb539745c418786434d117982d6cb : SUCCESS in 2h 13m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f43bdf0e2b1546d982ea778176cfc496 : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":4},{"id":"a3e40f291e1c193467c77a538dd874407e3c57af","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 14:37:08.000000000","message":"Patch Set 4:\n\nrecheck  cinder-tempest-plugin-lvm-lio-barbican failed on cinder_tempest_plugin.api.volume.test_volume_backup.VolumesBackupsTest.test_incr_backup_create_and_restore_to_an_existing_volume: tempest.exceptions.VolumeRestoreErrorException: Volume e68236e1-3ca4-4e2f-a5e8-5750a6e8197a failed to restore and is in ERROR status","accounts_in_message":[],"_revision_number":4},{"id":"a999fb8d8b4902a4fcb291851338984642b4a818","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-04 14:37:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fed70dbfd7830eb07c51ac6def7056f1d8045ee5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-04 15:41:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/95/763195/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a5b1e9c/ : FAILURE in 2h 19m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/95/763195/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/fdca333/ : SUCCESS in 2h 17m 44s","accounts_in_message":[],"_revision_number":4},{"id":"30de0afe34743982b977ef66b62893eda41e26a9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-04 16:51:44.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cdd87e0e9d5041d8b53db2fc572a8fbe : RETRY_LIMIT in 1m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/6d3586f45d164c6ea1978b3bd10d0f56 : TIMED_OUT in 2h 39m 02s","accounts_in_message":[],"_revision_number":4},{"id":"c36fd7574d4e40d19ef41251a55da51bcb3935ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-04 16:57:43.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"dc6ee7c032909b449523482fe1f2951b57161046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:47:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07383f6646e74122b8edcc24b1930647 : SUCCESS in 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc71cdcd48c4a5886b1ad89076bf16b : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a5ff5eeffcf4d1e8243fa14ed074751 : SUCCESS in 14m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38db26e7c50c4827a7c3fda9a339ac51 : SUCCESS in 11m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04e3a8b4cdc249578a9301a84733084c : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d083d84228432cae16eb3306708338 : SUCCESS in 15m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/f2b13c7e43684b5cb7acb3e090b0d953 : SUCCESS in 59m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd4fb2397a8e43e6a025e7eebe597bf6 : FAILURE in 1h 10m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2b60681352f44f658b7e4297f3c297a7 : SUCCESS in 15m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4ea6b0b0860c43fda4c7f732ce73aa06 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66be8d798aa445f4950e61bb132dceba : SUCCESS in 12m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c228e895eee0445cb126e6ebbb217192 : SUCCESS in 14m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b28b28e3ca0c4870911822d172f98785 : SUCCESS in 56m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/da61c643b4d14a4db4d38d2e4d276d24 : SUCCESS in 7m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13ca7274d8e24f61b6cec5866d18bba1 : SUCCESS in 1h 17m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/92cf3d53f5ae475eb68463b93b64f140 : FAILURE in 1h 32m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d5f55a99ab94f1ba3c555d88bbec012 : SUCCESS in 1h 58m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e53ea550bc940a582402d7151f1de71 : SUCCESS in 1h 33m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9454c622f36d4b0b81fb1f207e8c7689 : SUCCESS in 1h 26m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b8c16db67d5d47b0813ac49fca612c0a : SUCCESS in 1h 32m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ebe4efbc9cee472d95fad1761dee1d36 : SUCCESS in 1h 58m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/decaa589776f4d55890ad98876187964 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":4},{"id":"49dbfdc3ced6fac5a55209fcbd8e554f05cd46ef","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-04 17:52:36.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fa921d9911f8bc85fa12ca49de44cd9dda6e3ed9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-04 17:53:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"edb65299f6fe5283dd87d903d65ce1b0da540a3d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 18:17:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/4/check/dsvm-tempest-cisco-zonemanager-job/0ec5a07/ : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":4},{"id":"d9f3607ed6d09de48488b750a93d59a89cd7b717","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 18:43:30.000000000","message":"Patch Set 4:\n\ntempest-integrated-storage failure was: tearDownClass (tempest.api.volume.test_volumes_snapshots.VolumesSnapshotTestJSON) Volume 81d7f710-4aac-4223-bcc6-8ca69ba3e3d1 failed to delete and is in error_deleting status","accounts_in_message":[],"_revision_number":4},{"id":"25f5757bcaf5690ae3f44ca57697a4f54486ba66","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-04 20:20:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/95/763195/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0673d9a/ : SUCCESS in 2h 16m 09s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/95/763195/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f3f5169/ : POST_FAILURE in 2h 07m 49s","accounts_in_message":[],"_revision_number":4},{"id":"7c3e79fd7e714d18aabd7f2249036709da2f3903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:47:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dd61fd3dd11493ea63207094599175c : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65694a8d553449a2bbe0b73b55879c05 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3db94026b3ba480f94d365d139ac5aed : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dea3ecf0ec034fadb36d79fc664507a4 : SUCCESS in 11m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad7ca6dc3d1e4da284948970bb724832 : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fc1b8e7dab0497499fa89450de91007 : SUCCESS in 22m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/2aae1eef3c624ceba017aeacb9c55768 : SUCCESS in 1h 03m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89bccad183334b04baace39ec1cc5b53 : SUCCESS in 1h 12m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/260db6823ae041b3af10bb2c3822c1af : SUCCESS in 14m 02s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac678f16c3774f0c93f904dbcbfcdbe6 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81cadbb7f750453d94f371eedf16640e : SUCCESS in 9m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc83b50d4c9e4b9bbd2c006abc3f8071 : SUCCESS in 8m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b6ee8b69770483cb59cc03926be80c2 : SUCCESS in 58m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a74c0131a2945b8b3f791a698e64990 : SUCCESS in 10m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bacf8e9997cd45cd95bd915b3d97c4e7 : SUCCESS in 1h 18m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/172685be3fb94ebaa535ab9b46445f19 : FAILURE in 1h 28m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6dc2cf4ed9ae43dbac9108516d4fb247 : FAILURE in 1h 12m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8d5b72f252c04323be428219b5478d60 : SUCCESS in 59m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e7998dcd4d184c2ba60db9f1ce400e56 : SUCCESS in 1h 31m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/935687de50a045cfab1fe739a2f398bb : SUCCESS in 1h 21m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46ef71ab2a3246f393103c3032712e3c : SUCCESS in 2h 13m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a11e4cbf870b40ff9b2ae5d03a99f459 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":4},{"id":"e7975469b35449b1afb083c9de65099e86e61bbb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 21:49:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9b060c1b6fa84c23e9badb82f7c631713afd93ff","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-04 21:49:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a0be288111097bfc8d65246f8d5242f609a9b998","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 21:58:20.000000000","message":"Patch Set 4:\n\ncinder-tempest-plugin-lvm-lio-barbican had 5 failures, none related to this patch:\n\n1 tempest.api.volume.test_volumes_snapshots.VolumesSnapshotTestJSON.test_snapshot_create_get_list_update_delete - snapshot fd1e1283-c48d-47ca-bca0-ae1c3c817765 failed to reach available status (current creating) within the required time (196 s).\n\n2 cinder_tempest_plugin.api.volume.test_create_from_image.VolumeAndVolumeTypeFromImageTest.test_create_from_image_with_volume_type_image_property - volume 1eb23d61-7646-4d55-96f9-d052c1e2e4bb failed to reach available status (current creating) within the required\n\n3 cinder_tempest_plugin.scenario.test_volume_encrypted.TestEncryptedCinderVolumes.test_boot_cloned_encrypted_volume - volume 097599b7-921d-4278-8df4-7508f69114e5 failed to reach available status (current downloading) within the required time (196 s)\n\n4 tempest.api.volume.test_volumes_snapshots.VolumesSnapshotTestJSON.test_snapshot_create_offline_delete_online - snapshot 12c5a13d-fcd5-4503-bc5c-2acb1afe1f5b failed to reach available status (current creating) within the required time (196 s)\n\n5 tearDownClass (cinder_tempest_plugin.api.volume.test_create_from_image.VolumeAndVolumeTypeFromImageTest) - Failed to delete volume 1eb23d61-7646-4d55-96f9-d052c1e2e4bb within the required time (196 s)","accounts_in_message":[],"_revision_number":4},{"id":"1d87d871c09fbf862285e290d362a06e97948fad","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 23:42:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/4/check/dsvm-tempest-cisco-zonemanager-job/7f334aa/ : SUCCESS in 1h 19m 54s","accounts_in_message":[],"_revision_number":4},{"id":"40160497be035555baef227047f36e9d8fcb8798","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-04 23:47:22.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerFlex/9976968/DellEMC_PowerFlex/476 : SUCCESS in 2h 40m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"a916c2bf720a59e703b5bdd73b5e605cf1277e3f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-05 00:08:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/95/763195/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/58342ad/ : POST_FAILURE in 1h 55m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/95/763195/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5f7e622/ : SUCCESS in 1h 50m 07s","accounts_in_message":[],"_revision_number":4},{"id":"4d0cf50abbb624f9d7cd759c800ab1ee77dc52b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 00:14:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f496997e7f944ab9b7bdde2f341bf740 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d43a2749bf496a92f2364c63b9a40c : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c73424a3b9c497c821d542c21d6d474 : SUCCESS in 11m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/357d049fac824378b8b0aa46db0f8c3c : SUCCESS in 9m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc6a026f191b4c1eab7c59c7ec646e43 : FAILURE in 12m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acdc4a7003b5411d8fefaff87f3c0926 : SUCCESS in 11m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/13e9cf404ed24df180eb26bcf208a17b : SUCCESS in 1h 07m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b4f1941dbf84f28839ad712a15f7663 : FAILURE in 34m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/682b4230448c45d7b70983add40b3a87 : SUCCESS in 16m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/348a331b2c0844d7a2229d46d8424311 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8858628db6274ac8b182c5ca4823fc80 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d19f0503ef3a495e975ef78653d84052 : SUCCESS in 8m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5513a7c574df40e59bf3d4dfe25a54ed : SUCCESS in 44m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ecd2f2de9ad44bfa60483a2c63a6741 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a79983d8e8b45359f427f1ce9f7cb33 : SUCCESS in 1h 24m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8c07c318be54448a2fcb439f32eede3 : FAILURE in 1h 34m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/acba7deff62048cd8b6d4aaeba2d5017 : FAILURE in 1h 33m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a852fd23733048d6b349225b09f603ee : SUCCESS in 1h 10m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/74c71a3cfa5f4a4383c984b51db2e0da : SUCCESS in 1h 42m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e1f7b9765dc3415ca02bbdc1587480fb : SUCCESS in 1h 20m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37dfa92d0b814bdc843de5f2c7d85762 : POST_FAILURE in 2h 23m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0e88145dc754ac38a6608ec00870cc7 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":4},{"id":"b6a5ea7f04ddb88aa1d534c559969d32d1999a8c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-05 01:52:08.000000000","message":"Patch Set 4:\n\nrecheck cinder-tempest-plugin-lvm-lio-barbican failure is cinder_tempest_plugin.api.volume.admin.test_volume_backup.VolumesBackupsTest.test_incremental_backup_respective_parents : backup 7d96672a-a9aa-4d48-8f59-2cd0fba78d7f failed to build and is in ERROR status\n\ntempest-slow-py3 had a POST_FAILURE, but all the tests passed.","accounts_in_message":[],"_revision_number":4},{"id":"d76b22464f6773c986a7934233c6d4d958c849be","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-05 01:52:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"381e921bad1c5992ca54df0f784583689711451f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-05 03:17:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/763195/4/check/dsvm-tempest-cisco-zonemanager-job/600ac4b/ : SUCCESS in 1h 20m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b6affac5a2f7301400bd297d631b414a470cd81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 03:58:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb31e39ba462450b859c7f44b2dc7bc1 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dad5442930a40228a88b3fb7108c546 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/757e7d91f9cd439abba88eb3f9558340 : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c43939c7d9f746bea28ec594f9784ccc : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f97751d535294f9682176b4dbccee73d : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a2bed887294535ae5c1fc8b9731660 : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/b29df1d9d93b484c80c9b4edd562d6a2 : SUCCESS in 1h 02m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/960e9cd2031f402ea70ede86dca0f095 : SUCCESS in 1h 08m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b9ab4b0891504aacb2240938ad00b736 : SUCCESS in 11m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/97fce993437b450c95f66153874ce318 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54a3adfffcf34f68a31088f1948aa2c7 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/84e7c26f9207458594bf7b75d3cdc7f3 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8b08f0c4d6e84dc0a4aba3bc124c11fa : SUCCESS in 38m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/989dbebab5424929becb2a6f8dae9d60 : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bb671831abd043f9a6c311f483e94783 : SUCCESS in 1h 18m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/15b3a58dacbe4cb2ba25dd2c51cc39a3 : FAILURE in 1h 57m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8ec577e9c8234c9d89a06e6151d67b5b : SUCCESS in 1h 37m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/73904cd792fc4a16938c59c6172fd5b9 : SUCCESS in 1h 05m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/533aad73476b46109c2472994240a882 : SUCCESS in 1h 21m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e1e5ad1f2b114e1db96012c63277622c : SUCCESS in 1h 28m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f11e3533db546528cad0432c3a270c7 : SUCCESS in 2h 02m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50a9282cff6045629d8f927d3d44740d : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":4},{"id":"c030c961587704c5aff9be76e7e8fa8e1e22ad7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 03:58:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8952392c8e2b89f4de3254267f54950e81e16fd9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-05 04:28:20.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/6f9e17897b8f4beaa48b7e8242570ff9 : TIMED_OUT in 3h 12m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/037fa46739e14890b1b6507ea6a369e4 : POST_FAILURE in 2h 20m 52s","accounts_in_message":[],"_revision_number":4},{"id":"4301899b5d98351b94b8f4ba4cf4bc7b260c8009","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-05 05:02:14.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/763195-4/PVME-iSCSI : FAILURE in 15:17.01\n* PVME-FC http://os-logs.tristero.net/dell-ci/763195-4/PVME-FC : FAILURE in 15:24.03","accounts_in_message":[],"_revision_number":4},{"id":"2c50f0172e6bcb1bb218b12ca9405938f521c594","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-05 05:17:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-iSCSI/7da230b/ : SUCCESS in 37m 32s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-iSCSI-IPv6/b970a4f/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-NFS/ec89baf/ : SUCCESS in 38m 48s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-NFS-IPv6/7cd83eb/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/95/763195/4/upstream-check/cinder-cDOT-FCP/a3be244/ : SUCCESS in 1h 18m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"70e66d49389a1f97775fc08a69176aa6a41fd70f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 05:29:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f881659dd56431c971d7e97eb523bd6 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c8a0a7358343e69f7469cf81a0322c : SUCCESS in 11m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/704bb04cc43b4dee9341953cf391be81 : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25ba0845ee7649d1a5b957bd3a83d204 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82d957a3aa3472ca5730999f3f29e91 : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/a2562b009fb7417ba49e4f4364e75c26 : SUCCESS in 1h 10m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d34009d7861f4d3bb7c4eee6665003c6 : SUCCESS in 1h 25m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1baccb968c1b4fe29abf32ee35ebc6d4 : SUCCESS in 1h 13m 20s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cdd09b8eb63d47c6be8a032c2e1947b0 : SUCCESS in 1h 11m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1414423f95ac446ca230af4044f997b1 : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":4},{"id":"6b4a1762fc7db234074792e5a9ff4a6c01078bca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 05:29:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9587e70687e2b39524749f14f2de652d3bfc8a3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 05:31:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeaf2b927fb64ff2ad6911c526effbbb : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4},{"id":"1ddf09f014e8f17986c1ef0c01f19d46d7b6f6bd","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-05 05:42:57.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerStore_iSCSI/98a6d58/DellEMC_PowerStore_iSCSI/515 : SUCCESS in 52m 43s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerStore_FC/efaaaec/DellEMC_PowerStore_FC/509 : SUCCESS in 53m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"d51e85940db0e44a9e1bd0b0ade7ebe04eb3703b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-05 05:45:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/95/763195/4/upstream-solidfire-check/cinder-solidfire-iSCSI/d532329/ : FAILURE in 1h 46m 45s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"8cda53517cbd1e86e759ca4598069d3fbbcb6e17","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-05 08:09:25.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_SC_FC/dd55bb9/DellEMC_SC_FC/399 : SUCCESS in 2h 09m 20s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_SC_iSCSI/11d5ebf/DellEMC_SC_iSCSI/396 : SUCCESS in 2h 11m 36s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"8d566d95e04b92f50d80ba7da39a8197154d19ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 13:46:34.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"6d36947461cca3999a2e38971b7079480f71111d","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-05 16:51:41.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/763195-4/Seagate-iSCSI : FAILURE in 54:23.26\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/763195-4/Seagate-FC : SUCCESS in 43:46.93","accounts_in_message":[],"_revision_number":4},{"id":"391c624a797ec20068549c85740b2a89cb666139","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-05 18:11:50.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_VNX_iSCSI/520fb07/DellEMC_VNX_iSCSI/484 : SUCCESS in 3h 41m 41s\n- DellEMC_VNX_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_VNX_FC/66ce5ff/DellEMC_VNX_FC/504 : FAILURE in 1h 31m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"dd67573a7953ecfd5cce04a37b0e856e90b6489d","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-05 18:23:21.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/763195-4/Lenovo-iSCSI : SUCCESS in 49:06.94\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/763195-4/Lenovo-FC : SUCCESS in 46:16.93","accounts_in_message":[],"_revision_number":4},{"id":"ae2b31fb0e24cb0aaaa3225765452771e480598e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 20:23:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"10ad463b80a8f6a000865a5b3cd76de533eaaea8","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-05 21:01:58.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/763195-4/HPMSA-iSCSI : FAILURE in 1:11:08\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/763195-4/HPMSA-FC : SUCCESS in 40:16.58","accounts_in_message":[],"_revision_number":4},{"id":"d2ee23abe39d7504573c6b83d07afd42e624a8b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-06 11:17:47.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerFlex/9002ef0/DellEMC_PowerFlex/487 : SUCCESS in 2h 41m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f8e537a615c7a93a70e255d80efe7b41e07f2534","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-06 13:04:57.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"b2939dbe5902c27c3a164045c80380c0c4d2fd76","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-06 16:17:06.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerMAX_FC/6b21a4a/DellEMC_PowerMAX_FC/536 : SUCCESS in 5h 16m 22s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/95/763195/4/check/DellEMC_PowerMAX_iSCSI/caef5f5/DellEMC_PowerMAX_iSCSI/501 : SUCCESS in 4h 30m 58s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"1fe945b0ce9c4f4662c4dd2936fba46c7c952b73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-06 19:27:31.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/95/763195/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9fc6fab79d10a66a8e388d22e7f0ccb736b22eea","revisions":{"8a534eb51f1f4bf27a3dcf92d1431d95cbb044a3":{"kind":"REWORK","_number":1,"created":"2020-11-18 14:41:53.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/95/763195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/763195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/763195/1"}}},"commit":{"parents":[{"commit":"30501b9ecd6d4ed8dd6405be01795d8634013b24","subject":"Merge \"Synology: Improve session expired error handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30501b9ecd6d4ed8dd6405be01795d8634013b24"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-18 14:37:56.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-18 14:37:56.000000000","tz":0},"subject":"Update secure RBAC check strings with descriptions","message":"Update secure RBAC check strings with descriptions\n\nThis commit attempts to clarify the intended usage of each persona.\nAfter reading the comment for each, it should be clear how and when to\napply them to specific resources. If not, then we should continue to\nimprove the definition and remove ambiguity.\n\nChange-Id: I7bad7b5ebaff3d191c0701a9e72ed4c6334ecaa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a534eb51f1f4bf27a3dcf92d1431d95cbb044a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a534eb51f1f4bf27a3dcf92d1431d95cbb044a3"}]},"branch":"refs/heads/master"},"b3e4b7608ca674cc83730ded332fddd2755d7af0":{"kind":"REWORK","_number":2,"created":"2020-11-18 14:55:37.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/95/763195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/763195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/763195/2"}}},"commit":{"parents":[{"commit":"30501b9ecd6d4ed8dd6405be01795d8634013b24","subject":"Merge \"Synology: Improve session expired error handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30501b9ecd6d4ed8dd6405be01795d8634013b24"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-18 14:37:56.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-18 14:51:08.000000000","tz":0},"subject":"Update secure RBAC check strings with descriptions","message":"Update secure RBAC check strings with descriptions\n\nThis commit attempts to clarify the intended usage of each persona.\nAfter reading the comment for each, it should be clear how and when to\napply them to specific resources. If not, then we should continue to\nimprove the definition and remove ambiguity.\n\nThis commit also adds a new composite check string for system\nadministrators and project members. This implements a more robust\nsecurity posture where system administrators are allowed to do things\nthat project member can do. Individual deployments can offload some of\nthe administrative burden to system members if desired by override the\ndefault SYSTEM_ADMIN_OR_PROJECT_MEMBER check string.\n\nChange-Id: I7bad7b5ebaff3d191c0701a9e72ed4c6334ecaa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3e4b7608ca674cc83730ded332fddd2755d7af0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3e4b7608ca674cc83730ded332fddd2755d7af0"}]},"branch":"refs/heads/master"},"889611b531b37532ad6229ff85f150ae38803832":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-16 16:02:33.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/95/763195/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/763195/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/763195/3"}}},"commit":{"parents":[{"commit":"e28e1a2f41f38bb300884b25c8ea6b333da9488b","subject":"Move trace methods from utils to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e28e1a2f41f38bb300884b25c8ea6b333da9488b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-18 14:37:56.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-16 16:01:22.000000000","tz":0},"subject":"Update secure RBAC check strings with descriptions","message":"Update secure RBAC check strings with descriptions\n\nThis commit attempts to clarify the intended usage of each persona.\nAfter reading the comment for each, it should be clear how and when to\napply them to specific resources. If not, then we should continue to\nimprove the definition and remove ambiguity.\n\nThis commit also adds a new composite check string for system\nadministrators and project members. This implements a more robust\nsecurity posture where system administrators are allowed to do things\nthat project member can do. Individual deployments can offload some of\nthe administrative burden to system members if desired by override the\ndefault SYSTEM_ADMIN_OR_PROJECT_MEMBER check string.\n\nChange-Id: I7bad7b5ebaff3d191c0701a9e72ed4c6334ecaa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/889611b531b37532ad6229ff85f150ae38803832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/889611b531b37532ad6229ff85f150ae38803832"}]},"branch":"refs/heads/master"},"9fc6fab79d10a66a8e388d22e7f0ccb736b22eea":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-17 17:48:52.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/95/763195/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/763195/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/763195/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/763195/4"}}},"commit":{"parents":[{"commit":"bdfcdb631c59405a81e3fd0739d7ebbef893d49c","subject":"Merge \"Use os-brick 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bdfcdb631c59405a81e3fd0739d7ebbef893d49c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-18 14:37:56.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-17 17:44:40.000000000","tz":0},"subject":"Update secure RBAC check strings with descriptions","message":"Update secure RBAC check strings with descriptions\n\nThis commit attempts to clarify the intended usage of each persona.\nAfter reading the comment for each, it should be clear how and when to\napply them to specific resources. If not, then we should continue to\nimprove the definition and remove ambiguity.\n\nThis commit also adds a new composite check string for system\nadministrators and project members. This implements a more robust\nsecurity posture where system administrators are allowed to do things\nthat project member can do. Individual deployments can offload some of\nthe administrative burden to system members if desired by override the\ndefault SYSTEM_ADMIN_OR_PROJECT_MEMBER check string.\n\nChange-Id: I7bad7b5ebaff3d191c0701a9e72ed4c6334ecaa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fc6fab79d10a66a8e388d22e7f0ccb736b22eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fc6fab79d10a66a8e388d22e7f0ccb736b22eea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
