)]}'
{"id":"openstack%2Fcinder~764275","triplet_id":"openstack%2Fcinder~stable%2Ftrain~I4383b691a8cc4aacb046332e418aafb88ba8ba56","project":"openstack/cinder","branch":"stable/train","topic":"bug/1891914","hashtags":[],"change_id":"I4383b691a8cc4aacb046332e418aafb88ba8ba56","subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","status":"MERGED","created":"2020-11-26 04:15:40.000000000","updated":"2021-03-31 07:34:46.000000000","submitted":"2021-03-31 07:26:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"764275-bug/1891914","meta_rev_id":"15974be64a5e5018e9410f8a94ab3eb72b4bfcaf","_number":764275,"virtual_id_number":764275,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 07:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":2,"date":"2021-03-29 15:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-30 15:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2021-03-29 09:08:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-30 15:37:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2021-03-29 09:08:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 06:01:52.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-26 06:04:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-26 09:37:35.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-11-26 13:58:07.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-30 01:00:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-21 20:13:11.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-21 21:38:56.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-01-22 04:35:43.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-14 18:27:57.000000000","updated_by":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"reviewer":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"state":"REVIEWER"},{"updated":"2021-03-15 06:27:22.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-03-16 11:03:42.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-16 11:29:37.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-03-16 17:48:51.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-17 04:54:16.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-03-19 15:48:03.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-19 17:34:18.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-03-23 20:40:55.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-03-25 11:56:08.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-29 15:52:06.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"a36b66c563a4f40e6069de1289df34c94809fe98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-11-26 04:15:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"704c05cfeba9cb68dc73ac7d9ef7702450763800","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-11-26 04:17:54.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\nThis must wait Ussuri\u0027s backport to be merged first, also \"cherry picked from commit\" must be included before removing W-1.","accounts_in_message":[],"_revision_number":1},{"id":"395b1d8f94be6bba31a521d9daabc15f35b878fa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-26 06:01:52.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/1/check/tempest-dsvm-full-bionic-py3/1346cd5/ : SUCCESS in 1h 26m 54s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/1/check/devstack-plugin-ceph-dsvm-bionic-py3/b14391a/ : SUCCESS in 1h 31m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b44c199ce3a71939a60c04da1362cdd844b9ca3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 06:04:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89a01b2cb3c848e38a93629c3e50c35a : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2cbc30c6cfd4facb8f88799c96a349b : SUCCESS in 7m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcc553454f8c4287b822f58d34c343c3 : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89d15f6b264547e6b372baa2968a0b6b : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afee23fa7f9d421db39f57497993c89f : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c44d4b4aa64317a40d6c3929f1bf79 : SUCCESS in 12m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/335ff0afdae04cfab469d3f35ff4e251 : SUCCESS in 1h 40m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4e186211ca44b749630597b733a6fff : SUCCESS in 1h 04m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/7d81791203644eb9ab28ead22cdf4614 : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4ebd5e537374c98a10a95e0b91d136b : SUCCESS in 1h 18m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6db61fbeddbc41e79ebd72b476ee3c29 : SUCCESS in 6m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81df3d7a0cf8487086da38a7d6af4062 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5d15079022874ecc8d2204f085e8d99c : SUCCESS in 5m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/88c577cef70f4dd8b3e2e2ad93f506e4 : FAILURE in 44m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bbccad5c7e0f4a62ad6c9695c5e69792 : SUCCESS in 4m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8cce90abcea94597a1fb428fe4b7929f : POST_FAILURE in 1h 18m 41s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/57b177cafc6743b6a5083e2569e12556 : SUCCESS in 6m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7ab80486ee66477a97ddaaf2f9e334fc : FAILURE in 1h 41m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6cb51fc7e1bd452f8619c2e3597a6e33 : SUCCESS in 1h 25m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f708ff70d85f4580a109b50eba6d5d37 : SUCCESS in 1h 30m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab96ee40e9ba44babcb1f29d274a3429 : SUCCESS in 1h 29m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a283eeee5dec4f53be9c9fd0e329078d : SUCCESS in 1h 43m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bff4fa820a1744febd042a1eb636a78d : SUCCESS in 57m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/607bc6fbf7024348b098a08a3a77833e : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"12d9c8ee9912cd8a11b259661bf26da80a88061e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-11-26 09:37:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/75/764275/1/check/dsvm-tempest-cisco-zonemanager-job/9940e7c/ : FAILURE in 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8ce6f4200ba37f1b772761d15b729deb010de930","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-26 13:58:07.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-1/ns5_nfs : FAILURE in 9h 35m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-1/ns5_iscsi : FAILURE in 9h 35m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d30fc69178e5eb129b26578f3ea6b09c8e08b183","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-26 17:22:02.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-1/ns5_nfs : FAILURE in 3h 9m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-1/ns5_iscsi : SUCCESS in 3h 3m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c2571843893bf448cc9a6372122dfb83de6902f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-26 23:10:55.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-1/ns5_nfs : SUCCESS in 3h 4m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-1/ns5_iscsi : SUCCESS in 2h 41m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b6961fad59426ae02b9d004b6db3291784ac8613","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-11-30 01:00:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/75/764275/1/20201130075807/NEC-FC : SUCCESS in 1h 53m 54s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/75/764275/1/20201130075807/NEC-ISCSI : SUCCESS in 2h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"aa5fad0f055174f6d40eac81c7bdc044701a0e2a","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-01-21 20:12:26.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9aceda20514f481a97f674b882c869f274b21a6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-21 20:13:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70e92842c44603814410a31fea305552a861d946","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-21 21:38:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/764275/1/iscsi/ : SUCCESS in 1h 09m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/764275/1/smb/ : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"97a5b43d9140d06b4c7e9f154782d90a6690c00d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-21 23:23:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/75/764275/1/check/dsvm-tempest-cisco-zonemanager-job/9b4f345/ : FAILURE in 19m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7985ceb600cbc4df6ac5ed567d23f59fe95287af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 02:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd8adf02effa4e1c88a7d72ecc0ec256 : SUCCESS in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ee09a486944ce8a6945b705027ca6c : SUCCESS in 10m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/848f88b55a2b4463b7e0496dc2bc0c31 : SUCCESS in 10m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c4e96a6b1f0431985c42f57dc537596 : SUCCESS in 12m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea0685dcb7484c33b3593a6c339ae9ba : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1950c45551dc4f7d97c08ea36b807c35 : SUCCESS in 12m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/61df8225dafe4dabbc78d5ce831477a6 : SUCCESS in 1h 35m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4bf6497d18894b91bd1e979c22bea8d9 : SUCCESS in 55m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/8aa2794155d44adcaab8265eab921e1d : SUCCESS in 1h 09m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2210a90c1174bb09a769abe2bf719bc : SUCCESS in 1h 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1140d2aae9e4b40a29ca6515fc8c3e1 : SUCCESS in 6m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c07d5dc7e960449797f6ac0c08b88675 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b24dc9ff3ed4fa090552fc1828c9ceb : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8733e5b2a9ed49bca4ad1e0549ea4034 : FAILURE in 18m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/421c67d401a548a7b685af70d3ee8f4f : SUCCESS in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/93a557325e6f4d55a5ba57f72a7f4d7f : SUCCESS in 1h 23m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1c9051593f084b0891c2b1ebd886e3b3 : SUCCESS in 5m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/79419be0d37e4f7f80721545dc4e09f9 : SUCCESS in 1h 34m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ed7526a26f7c4524bba8d212fabe7c98 : SUCCESS in 1h 12m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/561f5cafa62747b592cc3813fefe0965 : SUCCESS in 1h 32m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c96365f422834256afa92d269e98c871 : SUCCESS in 1h 25m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/41d767bf56e64d429aa36a128768ef62 : SUCCESS in 1h 54m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/96c4e0a43fb84e4fa166cd2bc58daeb8 : SUCCESS in 57m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c047897217e94d07a22faaf596e75cbe : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4c9bf78b32a872ab63ef577519a6568863e97db2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-22 03:58:16.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/1/check/tempest-dsvm-full-bionic-py3/b178c85/ : SUCCESS in 1h 45m 21s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/02e9abfd92ef48d983c5df9e34d52fdc : RETRY_LIMIT in 47m 35s","accounts_in_message":[],"_revision_number":1},{"id":"16c5e36f458034e786951158569f87849f60a665","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-22 04:35:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/75/764275/1/upstream-solidfire-check/cinder-solidfire-iSCSI/30d9208/ : FAILURE in 2h 05m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"5b10d8bd5b2bbb7d71aecbc9a0e9c1459198c6d2","author":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"date":"2021-03-14 18:27:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3420704406332ad7e8b7cd7f0c8ee117aad8f20d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-03-15 06:27:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nEverything looks good but need to include hashes of victoria and ussuri in commit msg","accounts_in_message":[],"_revision_number":1},{"id":"6bb88e0c2c31c20e996f4e8c0ea02167a655c78e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-16 10:57:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2f33c3fa78cd80c12b9a2db4d29d57a6a82689fc","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-16 10:58:34.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks Rajat, I´ve just fixed the commit message with the correct hashes.","accounts_in_message":[],"_revision_number":2},{"id":"eaeb07423428d33b4d4a8f46f623572bcf54fb6b","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-16 11:03:42.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/75/764275/2/check/dsvm-tempest-TOYOU-ACS5000-driver/9bea009 : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f916b9488c93382969815aae78a5e27179778552","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-16 11:29:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d6ebd732b1e49475a4b585c59ac01405ac71f48f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 12:37:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28d78e4e16344e9e8932ac75c73ecde2 : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003e45023da7488f8fd7fad89f756b36 : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ecf6413352664129af0eac8bddc88d65 : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2994aae3e5e4d8bb4384a8df94d2686 : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f36ba9d1c54946c7801ce57dad4e9ae5 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c7352da6814bb4955e928446333ecd : SUCCESS in 12m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8fb22d3bb8724ab5a4ad57e6b368fd67 : SUCCESS in 1h 35m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3338c675fe9642cb80dce0d48e1a6d9f : SUCCESS in 52m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/608f823e1c5c425b8fafa3d3d584ec0d : SUCCESS in 1h 01m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/500dd4f5333840468adfd24526171a3b : SUCCESS in 1h 06m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/581b96a36f9941a4a9b6acefd24d77ed : SUCCESS in 6m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c751bdae2f0541b28d28c0ccac444989 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/51e364d4eb3f4d0cb4b3944fd04e4d2d : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/45114df17a1f4099ae7c778d01d3e369 : FAILURE in 18m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9d1bb037d4d147a48466c043d31fc580 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c362c67357b4864b0031b318f4d1735 : SUCCESS in 1h 13m 16s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c8c55d26bcd54e9d9f5257263a098a9b : SUCCESS in 5m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5728b1b736245d8a4a664008b23aea8 : FAILURE in 1h 25m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e17c767be8d40b1bc4519d70d9cc8f0 : SUCCESS in 1h 09m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aeaacd427bb246c4b96523046eec5fc9 : SUCCESS in 1h 21m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1edfdf5474ab4f6e8303f1db74930c09 : SUCCESS in 1h 11m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca5b34d2856e4d6fa60d9cae052be099 : SUCCESS in 1h 31m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f2c471de2e1e4716b9fbba678fbda28c : SUCCESS in 1h 01m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9da1ce08fb741c6a8127b66a827c434 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4cb0944e0a4d1fae91d5ad3ac80cf2d88d0bb4cd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-16 13:05:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2da11e2e77c60285cc4cc41132386748ceda4e21","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-16 14:01:41.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/2/check/tempest-dsvm-full-bionic-py3/2e340a2/ : SUCCESS in 2h 07m 19s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/2/check/devstack-plugin-ceph-dsvm-bionic-py3/ac3e2fa/ : SUCCESS in 1h 53m 16s","accounts_in_message":[],"_revision_number":2},{"id":"61b7c3f589640e6d5a6a019669f347acf4ecbabf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-16 15:03:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/75/764275/2/check/dsvm-tempest-cisco-zonemanager-job/efcb135/ : FAILURE in 20m 45s","accounts_in_message":[],"_revision_number":2},{"id":"7dbfb863d0fa4035b816a4527b867fa72d2d364d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 15:18:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66aa43c1392d4978ae890cc0b9202ea7 : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b92c7d7025da4a06a9a26b048e746246 : SUCCESS in 9m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/015451865cf74781ad8922c57b53e0e1 : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aba3eaeba9854393bab9afd32d8d29ae : SUCCESS in 10m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e49332e34b68498484e51ccb8fe901b2 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb4ac8ea9a243589d59521579066be2 : SUCCESS in 14m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1cee6f3e439a4579bac1798d1a552631 : SUCCESS in 1h 40m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c04f5608028f453895ee4dba4650ff68 : SUCCESS in 57m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/07988d3fb7ed4c3aa68f6a2c2ab24f26 : SUCCESS in 1h 19m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f7cef2919494338a2a75b6920ec8bb9 : SUCCESS in 1h 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9247269f5e6b4e1493b1fbcf6922e4c1 : SUCCESS in 7m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8046251c4f404943a093df4e50a930e8 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a228a40965374d588fe3f1368a8411e7 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f10a9ddaf0324732bf19c3ea5b4ad4a0 : FAILURE in 19m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c5a2d9f2a894bc58871e21b3c677b8b : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa6c560dfe1448c19054a9386bedd372 : SUCCESS in 1h 20m 22s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1542087d4899450fa2672cd76705e690 : SUCCESS in 6m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a6524112ccb8454cba08bf7b18383443 : SUCCESS in 1h 24m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/27545d345a9545cab6500f7828f687eb : SUCCESS in 1h 16m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d58776a70f994def9301ec9e1ce94b91 : SUCCESS in 1h 22m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/183dfea2ec2d46b7a1bf6a255ca54105 : SUCCESS in 1h 28m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/663b137df93340d6930ebd3fd3099230 : SUCCESS in 2h 08m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/96429d7c4fe14ec4913e81eefbd31204 : SUCCESS in 59m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/699d08e887424739a91f24f8efae25b5 : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f22f3de76fe348ef025dd44e27cb04ef2601bd04","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-16 16:19:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/75/764275/2/upstream-solidfire-check/cinder-solidfire-iSCSI/3a391c3/ : FAILURE in 59m 38s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"aaa6a827a56a07d08ac76c34820ecb8f166da057","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-16 17:48:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/75/764275/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7d2dd23/ : RETRY_LIMIT in 5m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/75/764275/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5f90e7f/ : RETRY_LIMIT in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d6f9c0402fe7f2a2ecc2c8bc87639eef5c8dc089","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-16 23:19:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/75/764275/2/check/dsvm-tempest-zadara-driver/fd0596c : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce61a85172c4ab1ef3c57b22d9a54691918aaa18","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-17 01:09:30.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-2/ns5_nfs : SUCCESS in 5h 37m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-2/ns5_iscsi : SUCCESS in 5h 19m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"2110c84ad38fea2662766fd5236d578f15151050","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-17 04:54:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8fa69e93ebcb5a5cde0e40701eb25224054ec8fa","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-18 10:24:37.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/75/764275/2/20210318180338/NEC-ISCSI : FAILURE in 1h 20m 27s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/75/764275/2/20210318180338/NEC-FC : FAILURE in 1h 20m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e1af517a3b3d4d3179370b1a40e7cb950b0f9a48","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-19 15:48:03.000000000","message":"Patch Set 2: Code-Review+2\n\nAppropriate backport.","accounts_in_message":[],"_revision_number":2},{"id":"c2506c8ce9509883b7ee708a9bd2acd6ab5be948","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-19 17:34:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca9e151ceeaaccabb16ddf03b4328006d4f7a843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-23 18:13:17.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6814391f3bc2bdfb1ee48ed6d3c08b23c4cace49","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-23 18:14:20.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"4728c0d240a4b80960e2a5e630548c5aea0b6f09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-23 18:16:17.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c7dd7b079123010058b6acb936db61d1ab9a13d6","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-23 18:16:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5e7d2f31110e9c9a59a6637b4f50d06e72d4f39e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-23 18:17:06.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0333e3316a05d33d03391b8d0b83d337fc259023","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-23 18:17:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"ea52643c2e51bd2faf5999ce67b1026b7d2cf645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-23 18:18:06.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e0fd352afcc8cfdcbf2d9098721b21958bcad5c5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-23 18:18:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"be97efdba81363ad106c011bc0d2cf79c6b52bd2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-23 18:19:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"0520c81f9e9591f7f2699a1420889e42e81085b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 20:23:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07ed7d70fbae424d9d3db70e4795b55a : SUCCESS in 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894e455946ec4238ac68f90c35659b7e : SUCCESS in 9m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39706c70f4c245b78c3dc73b6191dca9 : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cb843c46eac4e4aa40ca10e5b200710 : SUCCESS in 10m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e6243a4b802456b9c39e9a23e06f81f : SUCCESS in 15m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9770956812ba4c65879f321e5f5b89c5 : SUCCESS in 13m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3bee867f3cf54abc8881d686629ee65a : SUCCESS in 2h 02m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cf70050a62224bef912aee66f08fc682 : SUCCESS in 57m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/e2b2f9dc1cdd44d093fa6b6aa60f885d : SUCCESS in 1h 00m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95e75f9db7a647c0b95b9095489311f9 : SUCCESS in 1h 16m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/863ed606cec24ba6ae4d103b0409e58b : SUCCESS in 7m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5c881c87962f44a4b3623c045a100f68 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d9dd1b8b0b1d48ffbe09028c3455146b : SUCCESS in 7m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/32d24059ab0041738b5a6ad0b38b14b7 : FAILURE in 16m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4961ac3d3fea490f81769b0ee1ddc32f : SUCCESS in 8m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/978fcad1a0784a3a8b14bcadff0a3a79 : FAILURE in 1h 20m 00s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/36165d5e063040848db36d4d7120fb5c : SUCCESS in 7m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae6ea47e54444bd982c072be7bcb222b : SUCCESS in 1h 35m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/23005c65bc8e445697a2daa3cb52639f : SUCCESS in 1h 20m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6afd0f03c5d0467abe461f13ed4f876e : SUCCESS in 1h 32m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3c26c1bb054b4f4a8a544cd623fa80b9 : SUCCESS in 1h 42m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f60ed277f83a4b1eaeca40d3a17f8d5d : SUCCESS in 2h 00m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/69209bb11dc14b76a106cba9fcce69ed : SUCCESS in 1h 01m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf5ad987fad24178893a92166dc3e34c : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":6},{"id":"8d2229891ab309c2be97624b4a73de9be536dd2b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-23 20:36:35.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"fc8e5985301e4a179b7b29c2d861c725ef6fc984","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-03-23 20:40:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-75-764275-6/1616530999 : FAILURE in 17m 35s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":6},{"id":"d175ff53547d58fcc598a3990f86a1912bb18a8e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-23 20:50:56.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/75/764275/6/check/dsvm-tempest-cisco-zonemanager-job/6d312e7/ : FAILURE in 22m 02s","accounts_in_message":[],"_revision_number":6},{"id":"b3f3e8a0c94a4b193352c8545a50343a8a65a007","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-23 21:51:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/75/764275/6/upstream-solidfire-check/cinder-solidfire-iSCSI/b032745/ : FAILURE in 1h 27m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"b9a7a0a5a441c6a4e84877e0bc4d2c31062e5541","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-23 22:33:05.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/6/check/tempest-dsvm-full-bionic-py3/4612930/ : SUCCESS in 1h 50m 53s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/6/check/devstack-plugin-ceph-dsvm-bionic-py3/c674833/ : SUCCESS in 1h 51m 44s","accounts_in_message":[],"_revision_number":6},{"id":"4eeadbbebc560b4ebb9cc558b25773541e93593d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-03-24 04:19:18.000000000","message":"Patch Set 6: Code-Review+1\n\nMerged in ussuri, looks appropriate to backport.","accounts_in_message":[],"_revision_number":6},{"id":"301e6c1cef4c6b29a3362e9c644dc2e847cda8dd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-24 05:10:05.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-3/ns5_nfs : SUCCESS in 4h 41m 16s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-3/ns5_iscsi : SUCCESS in 4h 28m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8617966ad4dd9c7082ac7d18e7cfc40a586cbbe2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-24 06:04:16.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-4/ns5_nfs : SUCCESS in 4h 35m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-4/ns5_iscsi : SUCCESS in 4h 26m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"70d9acabf83fcd5028e98aa45020a9c6be4d055d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-24 07:08:25.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-5/ns5_nfs : SUCCESS in 4h 39m 37s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-5/ns5_iscsi : SUCCESS in 4h 32m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"200fa484b54d6960b4f04ec0f00b39ce4303be04","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-24 08:06:03.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-6/ns5_nfs : SUCCESS in 4h 37m 9s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-6/ns5_iscsi : SUCCESS in 4h 24m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"aafb5a30fa7063f5bb36a2ffdba5a9a14956684c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-24 08:34:06.000000000","message":"Patch Set 6: -Code-Review\n\nuh, a rebase is now needed (maybe those backporta should be stacked)","accounts_in_message":[],"_revision_number":6},{"id":"c0d7d39dcb9b9da26b483e1e113ae97a4c906392","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-24 09:54:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/75/764275/6/check/dsvm-tempest-zadara-driver/f699da6 : SUCCESS in 1h 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b199cd2aab19c46a5b5025627111b4c8a6d27003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-24 16:19:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"66125b99c7f2fe035d95b0bcad55ee94419bcfc1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-24 16:20:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"a743ffc4530d17e972356f7d1c832e460d05bc62","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-24 16:22:20.000000000","message":"Patch Set 7: Code-Review+2\n\nThis is good to go once the check passes.","accounts_in_message":[],"_revision_number":7},{"id":"b06e2cc18817e7583c29f675aa4dfdf922f247a0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-24 16:47:09.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/75/764275/7/check/dsvm-tempest-cisco-zonemanager-job/2a0fcc7/ : FAILURE in 21m 42s","accounts_in_message":[],"_revision_number":7},{"id":"2fcf1d7970995fc851c328ed3c5290885dc04b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 18:45:46.000000000","message":"Patch Set 7: Verified-1\n\n(44 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/495cbb3cb0c0401aa4c433d47b052669 : FAILURE in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a10726f0890471db0134a6da6ce277a : FAILURE in 13m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46485311415148db86f8991bbd7529f5 : FAILURE in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ebe35adc6aa409ebb069439b1d81ba5 : FAILURE in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f19088cb362e4f39bf0e977f19eb4ef6 : FAILURE in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0fd85cbdf74e2ca90bc76b2d364ccf : FAILURE in 5m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dde4f1857955417aac39962cc45448dd : SUCCESS in 1h 21m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b929c5cb551b4bfcaa02599367db26b4 : SUCCESS in 49m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7073cd77005e4d149df5f700c43f70be : SUCCESS in 51m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e1be0b4faa1407384dd039a9d32f58a : SUCCESS in 1h 05m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70482aac5bea47989ae91261acc5d2a2 : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4608ada465994b939b9017b69d4ad4a4 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1cc3039cfad4579a82247b1dda9b832 : SUCCESS in 7m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2872a7fc5e51485199cdef015f9032ec : FAILURE in 23m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc7d458c4ead472683dc3d30390f3417 : FAILURE in 8m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b762a0b8af43465098ab02b4c96458b1 : FAILURE in 1h 12m 44s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/77942d52c14e4d90a612ac5405463ca8 : SUCCESS in 6m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/acc3303dca30421fb3afdf3d5b556215 : SUCCESS in 1h 34m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6725df83b14347f685d430f2fd5e2b5e : SUCCESS in 1h 09m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/99751440d5624317b89d6379e9bec9e6 : SUCCESS in 1h 25m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1493d93f7ecc4eef83c5733f5e4c6b65 : SUCCESS in 1h 33m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/180bf89708b34152ad5add45474e2d90 : SUCCESS in 1h 53m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a10b4c789f8f433faf58b5bf715a58e3 : FAILURE in 39m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43a9ef05f05b4664b43ffaa9084bc97a : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":7},{"id":"cb9417be3f2ab99cb59e08c3bcf0404e62b76d7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-24 22:34:03.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"c124ee0b967ee5fd1e2d7798b81b142a8274c9fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-24 22:37:35.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"08883d4f78404bf9196f02c170cad4e603c80829","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-03-24 22:47:25.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent, and commit message as Patch Set 8.","accounts_in_message":[],"_revision_number":9},{"id":"86e1b0e286db16e15d6f2d92e0a7abf0855b821c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-24 22:48:37.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"df4dfed2e05d88dbe197ecf3120147f50ec3b24f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-24 23:14:35.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/75/764275/9/check/dsvm-tempest-cisco-zonemanager-job/b98b458/ : FAILURE in 21m 48s","accounts_in_message":[],"_revision_number":9},{"id":"b5a9a74538d21c899eee495a6da28b0bf46926f2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-24 23:30:56.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-7/ns5_nfs : SUCCESS in 4h 44m 46s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-7/ns5_iscsi : SUCCESS in 4h 43m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"ba6f292374c3a6b2c26bf4705185637d86f7c60e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 00:47:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8f9423296ba48299019478a3d05f667 : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8daba5e856164a979790eae77e08a079 : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bbcc3675deb4c0a8357263377eb4ff4 : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cf86d9a70704b7d9ce948a342b4a26b : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7070ec41ea004ee8919efbd24deb2085 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9926e33bb0134bd4a415a35e968e035a : SUCCESS in 13m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/daa95d52c05a42109a8e7c297ed87189 : SUCCESS in 1h 28m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/56bc92a86dff41d9956e29d989188826 : SUCCESS in 53m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/b05d64bb03604837a0e0263a309d62a4 : SUCCESS in 56m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60a878caa1d14f58ba8bd96f3a996867 : SUCCESS in 1h 07m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/340be7adf52c4a33a998d9cc81e73e06 : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b658e99e40834008b02efc33cb86ee78 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4b1b35d484a4043a1bc8fdf5d639a86 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6530547ee62e4b72b541c26c2e0adb5a : FAILURE in 18m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d6fdbb5875764eb497e0bf7be3968c1b : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4d35bfed5fce4e18b22de4b22f6acaef : FAILURE in 1h 24m 38s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/17d1433445044731ab03eda73cc6813f : SUCCESS in 5m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5fbb4c4a9d04c25b728f9254ba07bf4 : SUCCESS in 1h 25m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2f6eb0ce89c4469784690fd91d44dc3d : SUCCESS in 1h 02m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/49c1cd5e465f4b1f95c8eb24806ef663 : SUCCESS in 1h 25m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ef8b0358059c4c9abdaed190c16274ad : SUCCESS in 1h 19m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a26a63e18f4b4626a9c7daa5a5db1510 : SUCCESS in 1h 58m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2142849ed20446798e26597d1b12e8d9 : SUCCESS in 59m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/068d308c96374dfe95b9ae400760bd75 : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":9},{"id":"51e58361aec40a5fc1290a5ab53b916db1da81cd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-25 02:12:34.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/75/764275/9/upstream-solidfire-check/cinder-solidfire-iSCSI/74b0a28/ : FAILURE in 1h 22m 55s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"3ad2b8aa77de022121dd3461a9dbddf194c4f33c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-25 02:30:38.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/9/check/tempest-dsvm-full-bionic-py3/3de5056/ : SUCCESS in 1h 50m 14s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/75/764275/9/check/devstack-plugin-ceph-dsvm-bionic-py3/a089123/ : SUCCESS in 1h 49m 49s","accounts_in_message":[],"_revision_number":9},{"id":"7c941be67bb622d0b1a6f9734d088e2cb3134e2f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-25 07:08:34.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-9/ns5_nfs : SUCCESS in 5h 14m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-9/ns5_iscsi : SUCCESS in 4h 54m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"c037d00060af26c358797f9fcbc5648b1d676d2b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-25 07:19:38.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-8/ns5_nfs : SUCCESS in 5h 27m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764275-8/ns5_iscsi : SUCCESS in 5h 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"b6a0926344cee7798ca34a2af0fe2879bd61f624","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-25 10:01:22.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/75/764275/9/check/dsvm-tempest-zadara-driver/eef396f : SUCCESS in 1h 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1e8545d02676dd51430558a011361410f16327d6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-25 11:56:08.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/75/764275/9/check/iscsi-driver-nimble/0d3747f/0d3747f.html : FAILURE in 1h 17m 59s","accounts_in_message":[],"_revision_number":9},{"id":"1371e4f528b0fee46c420a2a8f906f02a06683ba","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 20:38:40.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/75/764275/9/20210327035721/NEC-ISCSI : FAILURE in 1h 40m 47s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/75/764275/9/20210327035721/NEC-FC : FAILURE in 1h 41m 17s","accounts_in_message":[],"_revision_number":9},{"id":"e4a4971d07bc2f55789d605afa6c22019097f372","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-03-29 09:08:36.000000000","message":"Patch Set 9: Code-Review+1 Review-Priority+1\n\nLGTM.\nsetting RP+1 since the deadline discussed for stable/train release is tomorrow (i.e. 30 March)","accounts_in_message":[],"_revision_number":9},{"id":"f3c7b13cfc582540b20d51789d1d0d230ce75a9d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-29 15:52:06.000000000","message":"Patch Set 9: Code-Review+2\n\nWhat makes this different from the backports to more recent branches is that stable/train supports Python 2.7.  That said, I don\u0027t see any python 2 incompatibilities in this change, and it\u0027s isolated to a single driver.","accounts_in_message":[],"_revision_number":9},{"id":"1ab8e5047959e7caf247c8ba87e49cc7557e962d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-30 15:37:07.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nLooks good to me.","accounts_in_message":[],"_revision_number":9},{"id":"dbe4a5ea9ba1ae9a2028dcdaa6c6597f10124010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 22:28:04.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"797bcf61d6bca2ee6c949c2175aa207d6c49202e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:25:58.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd557ae924f24b3998eeae56f121b38f : SUCCESS in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6c43e23fd64bd084b4e75ae3571891 : SUCCESS in 9m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74875303e0744f378588e835b16731fc : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6b64c2cd2694d1c8e7db6febc19d781 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91a5df36413d40d097f5610fc8d1617b : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654686d026544d91925e6ef244f1824d : SUCCESS in 15m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d26d261090cc43099486b5f8eb9ac2ad : SUCCESS in 1h 18m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6be6c192b9854b6db34085d7bb7482ae : SUCCESS in 55m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/3f6901b3a3354ed49005ab98117076f4 : SUCCESS in 1h 14m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/864f64e29c194ab7bf0fe21d8dfeb9fd : SUCCESS in 57m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9353c9581b0b47f28cffea9e6f65b519 : SUCCESS in 6m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f9c735ce8b0741e5aed9dc50324c0396 : SUCCESS in 1h 09m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/88beaa8f1e5747898185912bc7309ca0 : SUCCESS in 1h 17m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24233a9d92684c8b86ac2aa5649ddf94 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":9},{"id":"b9423bb178123ef2b4e40e814125c5f6298afca0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:26:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"15974be64a5e5018e9410f8a94ab3eb72b4bfcaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:34:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fec68ae4dd84336ba890cabda5570a6 : SUCCESS in 2m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fd144e05c514844b86cec5706c57196 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"f70bfbf7110fbf20f1e263ade02229c6837393bc","revisions":{"77ca45768e3b2fbe67c255fbf5568505324df204":{"kind":"REWORK","_number":1,"created":"2020-11-26 04:15:40.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/1"}}},"commit":{"parents":[{"commit":"88133fae4779887c4edf5914c72e6f1ed4c507b2","subject":"Merge \"PowerMax Train Docs - Upgrade improvement notification\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88133fae4779887c4edf5914c72e6f1ed4c507b2"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-11-26 04:15:20.000000000","tz":-180},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77ca45768e3b2fbe67c255fbf5568505324df204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77ca45768e3b2fbe67c255fbf5568505324df204"}]},"branch":"refs/heads/stable/train"},"db0c8ede59fccd1a8a13e36ce396e8429d8d9ed8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-16 10:57:03.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/2"}}},"commit":{"parents":[{"commit":"88133fae4779887c4edf5914c72e6f1ed4c507b2","subject":"Merge \"PowerMax Train Docs - Upgrade improvement notification\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88133fae4779887c4edf5914c72e6f1ed4c507b2"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz Silva","email":"sfernand@netapp.com","date":"2021-03-16 10:56:51.000000000","tz":0},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n(cherry picked from commit deb31a0c4d35ea6c479bfe466b780ed07a489276)\n(cherry picked from commit d73da73283c783f90e878c5fbd53e114e1918f0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db0c8ede59fccd1a8a13e36ce396e8429d8d9ed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db0c8ede59fccd1a8a13e36ce396e8429d8d9ed8"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"39575aff10d3dcc0a9f6e112ff869bd74ffff726":{"kind":"REWORK","_number":3,"created":"2021-03-23 18:13:17.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/3"}}},"commit":{"parents":[{"commit":"88133fae4779887c4edf5914c72e6f1ed4c507b2","subject":"Merge \"PowerMax Train Docs - Upgrade improvement notification\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88133fae4779887c4edf5914c72e6f1ed4c507b2"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz Silva","email":"sfernand@netapp.com","date":"2021-03-23 18:12:40.000000000","tz":0},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n(cherry picked from commit deb31a0c4d35ea6c479bfe466b780ed07a489276)\n(cherry picked from commit d73da73283c783f90e878c5fbd53e114e1918f0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39575aff10d3dcc0a9f6e112ff869bd74ffff726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39575aff10d3dcc0a9f6e112ff869bd74ffff726"}]},"branch":"refs/heads/stable/train"},"cf2c6fed704c679a5712bf0cc0bcbe792e10f444":{"kind":"REWORK","_number":4,"created":"2021-03-23 18:16:17.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/4"}}},"commit":{"parents":[{"commit":"88133fae4779887c4edf5914c72e6f1ed4c507b2","subject":"Merge \"PowerMax Train Docs - Upgrade improvement notification\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88133fae4779887c4edf5914c72e6f1ed4c507b2"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz Silva","email":"sfernand@netapp.com","date":"2021-03-23 18:16:04.000000000","tz":0},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n(cherry picked from commit deb31a0c4d35ea6c479bfe466b780ed07a489276)\n(cherry picked from commit d73da73283c783f90e878c5fbd53e114e1918f0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf2c6fed704c679a5712bf0cc0bcbe792e10f444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf2c6fed704c679a5712bf0cc0bcbe792e10f444"}]},"branch":"refs/heads/stable/train"},"64ebba6025ac3cd40de3fc427c96009f807f9508":{"kind":"REWORK","_number":5,"created":"2021-03-23 18:17:06.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/5"}}},"commit":{"parents":[{"commit":"88133fae4779887c4edf5914c72e6f1ed4c507b2","subject":"Merge \"PowerMax Train Docs - Upgrade improvement notification\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88133fae4779887c4edf5914c72e6f1ed4c507b2"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz Silva","email":"sfernand@netapp.com","date":"2021-03-23 18:17:02.000000000","tz":0},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n(cherry picked from commit deb31a0c4d35ea6c479bfe466b780ed07a489276)\n(cherry picked from commit d73da73283c783f90e878c5fbd53e114e1918f0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64ebba6025ac3cd40de3fc427c96009f807f9508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64ebba6025ac3cd40de3fc427c96009f807f9508"}]},"branch":"refs/heads/stable/train"},"7770891a06c41a1eea7dc856d2b62a3073d94586":{"kind":"REWORK","_number":6,"created":"2021-03-23 18:18:06.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/6"}}},"commit":{"parents":[{"commit":"88133fae4779887c4edf5914c72e6f1ed4c507b2","subject":"Merge \"PowerMax Train Docs - Upgrade improvement notification\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88133fae4779887c4edf5914c72e6f1ed4c507b2"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz Silva","email":"sfernand@netapp.com","date":"2021-03-23 18:17:59.000000000","tz":0},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n(cherry picked from commit deb31a0c4d35ea6c479bfe466b780ed07a489276)\n(cherry picked from commit d73da73283c783f90e878c5fbd53e114e1918f0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7770891a06c41a1eea7dc856d2b62a3073d94586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7770891a06c41a1eea7dc856d2b62a3073d94586"}]},"branch":"refs/heads/stable/train"},"fde8a22426f9ef0ac65514150bf485017553cadd":{"kind":"REWORK","_number":7,"created":"2021-03-24 16:19:49.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/7"}}},"commit":{"parents":[{"commit":"ae951681b6c9b517739e5ca6b51ab0772b2a12a6","subject":"Merge \"NetApp SolidFire: Fix clone and request timeout issues\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae951681b6c9b517739e5ca6b51ab0772b2a12a6"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2021-03-24 16:16:18.000000000","tz":0},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n(cherry picked from commit deb31a0c4d35ea6c479bfe466b780ed07a489276)\n(cherry picked from commit d73da73283c783f90e878c5fbd53e114e1918f0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fde8a22426f9ef0ac65514150bf485017553cadd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fde8a22426f9ef0ac65514150bf485017553cadd"}]},"branch":"refs/heads/stable/train"},"afad2855074bc3c01c60070c60baad3aba44dc47":{"kind":"REWORK","_number":8,"created":"2021-03-24 22:34:03.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/8"}}},"commit":{"parents":[{"commit":"ae951681b6c9b517739e5ca6b51ab0772b2a12a6","subject":"Merge \"NetApp SolidFire: Fix clone and request timeout issues\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae951681b6c9b517739e5ca6b51ab0772b2a12a6"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz Silva","email":"sfernand@netapp.com","date":"2021-03-24 22:33:53.000000000","tz":0},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n(cherry picked from commit deb31a0c4d35ea6c479bfe466b780ed07a489276)\n(cherry picked from commit d73da73283c783f90e878c5fbd53e114e1918f0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afad2855074bc3c01c60070c60baad3aba44dc47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afad2855074bc3c01c60070c60baad3aba44dc47"}]},"branch":"refs/heads/stable/train"},"f70bfbf7110fbf20f1e263ade02229c6837393bc":{"kind":"NO_CHANGE","_number":9,"created":"2021-03-24 22:47:25.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/75/764275/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/75/764275/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/75/764275/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/75/764275/9"}}},"commit":{"parents":[{"commit":"ae951681b6c9b517739e5ca6b51ab0772b2a12a6","subject":"Merge \"NetApp SolidFire: Fix clone and request timeout issues\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae951681b6c9b517739e5ca6b51ab0772b2a12a6"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-05 22:20:15.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2021-03-24 22:43:30.000000000","tz":0},"subject":"NetApp SolidFire: Fix duplicate volume when API response is lost","message":"NetApp SolidFire: Fix duplicate volume when API response is lost\n\nThe SolidFire driver retries API requests in case a connection\nerror occurrs. When network is unstable, there may be the\npossibility that the SolidFire backend successfully receive\nand process a create volume operation, but fail to deliver the\nresponse back to the driver.\n\nWhen this scenario occurrs, the SolidFire driver automatically\nresends the request, creating a second volume and leaving a\nduplicate unused. Although this doesn\u0027t affect\ndriver functionality at first moment (the volume id from the\ncluster is always correctly associated to cinder provider id),\nfurther operations may hit the unused volume, leading to\nunexpected hehavior.\n\nThis patch fixes this issue by:\n\n1. Checking if the volume name already exists in the\nbackend before trying to create it. Volume creation will\nraise a exception and abort in case of a volume is found.\n\n2. Checking for volume creation right after a read timeout is\ndetected, preventing invalid API calls.\n\n3. Adding option ´sf_volume_create_timeout´ to the SolidFire\ndriver, to allow users to set the appropriate timeout value for\ntheir environment.\n\nCloses-Bug: #1896112\nChange-Id: I4383b691a8cc4aacb046332e418aafb88ba8ba56\n(cherry picked from commit 42c92cc407d475751bc61c98445c9c0740a71496)\n(cherry picked from commit deb31a0c4d35ea6c479bfe466b780ed07a489276)\n(cherry picked from commit d73da73283c783f90e878c5fbd53e114e1918f0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f70bfbf7110fbf20f1e263ade02229c6837393bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f70bfbf7110fbf20f1e263ade02229c6837393bc"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
