)]}'
{"id":"openstack%2Fcinder~764942","triplet_id":"openstack%2Fcinder~stable%2Fstein~Ie330c76a5db0ea76d4fed5a6ae7b8736dadc8591","project":"openstack/cinder","branch":"stable/stein","topic":"bug/1898587","hashtags":[],"change_id":"Ie330c76a5db0ea76d4fed5a6ae7b8736dadc8591","subject":"NetApp SolidFire: Fix clone and request timeout issues","status":"ABANDONED","created":"2020-12-01 14:12:12.000000000","updated":"2023-05-03 14:29:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19fc3c6b133a94310d249d7e791f4875d25b63f8","_number":764942,"virtual_id_number":764942,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-13 15:11:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 14:13:35.000000000","updated_by":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-01 14:33:47.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-01 14:36:08.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-01 18:22:59.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-02 00:36:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 01:06:30.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-06 08:43:00.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-04-13 15:15:59.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 18:14:26.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"}],"messages":[{"id":"f947b31d6660572f14a4f0ac9a7e690398838864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-12-01 14:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a0f3bd9db143c7fcbb69dbbe1589c3a80232e40","author":{"_account_id":12032,"name":"Datera CI","email":"datera-openstack-ci@datera.io","username":"datera-ci","tags":["SERVICE_USER"]},"date":"2020-12-01 14:13:35.000000000","message":"Patch Set 1:\n\nBuild Unstable \n\n* datera-jenkins http://stkci.daterainc.com.s3-website-us-west-2.amazonaws.com/17ab8a438a4a19c6f4ab0a689163e8ad6db57543/index.html : UNSTABLE","accounts_in_message":[],"_revision_number":1},{"id":"f4624294e163efa85e4a410e93df546905f1afef","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2020-12-01 14:15:07.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\nThis must wait Train\u0027s backport to be merged first, also \"cherry picked from commit\" must be included before removing W-1.","accounts_in_message":[],"_revision_number":1},{"id":"9633077f10a1bd948c35d94cee4700d07491a644","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-01 14:33:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/42/764942/1/check/dsvm-tempest-cisco-zonemanager-job/10bbcb0/ : POST_FAILURE in 9m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7d1b51d9716eb3628f402bfe186d8813724b82c9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-01 14:36:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/764942/1/iscsi/ : FAILURE in 7m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/764942/1/smb/ : FAILURE in 8m 38s","accounts_in_message":[],"_revision_number":1},{"id":"df024d5dd396f9b63f0cd29bcabbdba3f00f4208","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-01 18:22:59.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/764942/1/check/tempest-dsvm-full-bionic-py3/6e1e332/ : SUCCESS in 1h 33m 59s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/764942/1/check/devstack-plugin-ceph-dsvm-bionic-py3/f9120ea/ : FAILURE in 1h 42m 12s","accounts_in_message":[],"_revision_number":1},{"id":"253f2824b77650d4a578960017c5a756a7657351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:36:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb6bc00fb9fa4ae7af65dd1eaf721229 : FAILURE in 29m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aedabed26f24a758b767fa34f58632e : SUCCESS in 8m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb7cc842f271410dbcb310d1281c545e : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba4fa0fdf9a94cb2b13b1d82835bd7e5 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec80e12431984f50bbf9d32aaa00f36d : SUCCESS in 9m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ea9c59391e724db0a6de1d0966557668 : FAILURE in 1h 16m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d166baec988a45b8978f314ad69a7139 : SUCCESS in 52m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ee607297a5c4af5ac9aaf6cebf1e350 : SUCCESS in 1h 16m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e00fe4c73229494abc9ad67178f47fdc : SUCCESS in 6m 50s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c72683304acf4097a634d93ba2db0ba2 : SUCCESS in 1h 29m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3496aa396f364d2e976bee208ca4e696 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5865132961734f26b1bbefa065f567de : SUCCESS in 5m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3f42e119a14e40a3a2f2bdb40659b6ef : FAILURE in 35m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1116fb7f59124693bdd6d472380257c5 : SUCCESS in 11m 08s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/99064f20215f4d49baaa9fc8f279c6a5 : SUCCESS in 1h 27m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/adcb15d870bd49d983c025b1aee796ba : SUCCESS in 5m 25s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e599a63130404ac385809565015bce93 : SUCCESS in 1h 09m 59s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f378a1a64734c6dbd3c543f04bf7e69 : SUCCESS in 1h 28m 00s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/0d5601776c38490589baf08a546225c7 : SUCCESS in 1h 43m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6aa7d445da0143068eb96f6fdb955d1a : SUCCESS in 2h 14m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bb6182ccd8d9450aad5dcabe382d8e1f : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"aa243755a7330f38bc3efdca764b88ddc2b26ddb","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-02 01:06:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/764942/1/20201202080908/NEC-FC : SUCCESS in 1h 48m 21s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/764942/1/20201202080908/NEC-ISCSI : SUCCESS in 1h 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a42a99991c6f3725008eeb7f9d7be4c226d2a9ea","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-04-06 08:43:00.000000000","message":"Patch Set 1:\n\nThis has a lower priority, but now that the train backport has merged, can you please update it according the same guidelines, with all the cherry-picks in place?\n\nYou may want to start updating also the backports to the older branches as well, even though they will need to wait until this merges.","accounts_in_message":[],"_revision_number":1},{"id":"f07b065beda0f9f31e1919616062b5100466964d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-04-13 13:06:05.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"10051da349b7b10024fb5a9c2f08bcdc5275560d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-04-13 13:10:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f2f0bf0313c4e5b1a79389e2901872842809c75e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-13 14:00:05.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/42/764942/3/check/dsvm-tempest-cisco-zonemanager-job/d236f07/ : FAILURE in 21m 20s","accounts_in_message":[],"_revision_number":3},{"id":"273dfc384b1716aa3801f8dd61505fd861d727c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 15:11:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a29a4a0c965c409e9d20c5353c6cc3a9 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d58bb2001e4a449abf7d81bfed09f0 : SUCCESS in 9m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa20b15c1b5e4edb92abaea6f79faded : SUCCESS in 10m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f122cef216149f6ab460b31c4505ccb : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2f56b83a0148adbecc850bf38c5fdc : SUCCESS in 10m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca8980b5f0fa42a28f45288183215429 : SUCCESS in 1h 32m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/22f3096e10c6421691ff0ea49e659eac : SUCCESS in 1h 03m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/467cf6b72f4c410280064fa4e9a018bd : SUCCESS in 1h 22m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea1e149d1ac547b190a64f04512a182f : SUCCESS in 7m 25s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf773ed0322343febb039db6d4405e94 : SUCCESS in 1h 18m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5998324cb2ff424d92bddde35fbcf93e : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de605ce9be284992ae8597d01d0dbb4e : SUCCESS in 6m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/24b52ef50cab44efbe37389d0458e1ea : FAILURE in 19m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/66335e8bcac5418ebf5eac1a2a8d0319 : SUCCESS in 7m 26s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc9632180049426e9217a50a506cacf7 : SUCCESS in 1h 08m 54s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2bee64ebe91048e9a96eea9891275432 : SUCCESS in 6m 49s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c7ad053e6aa54b9a87b0d8a045cb1865 : SUCCESS in 58m 31s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c799c7178b9a4660ab22170ae93fc827 : SUCCESS in 1h 13m 25s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/cb488e4be11c4fb6977dc0789a920727 : SUCCESS in 1h 17m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e36fa27ab4b64b2689c2024d48bf2637 : SUCCESS in 1h 56m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/080ca993ca9940c8ab87a59224df1f7f : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":3},{"id":"6206338f622ac29e81fc0cb81554f339c331c428","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-13 15:15:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/42/764942/3/check/iscsi-driver-primera/77e1f96/77e1f96.html : FAILURE in 1h 05m 30s","accounts_in_message":[],"_revision_number":3},{"id":"81a13facc7ae4c8801ccac9e8c7f48d6a9f407b7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-13 18:14:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/42/764942/3/upstream-solidfire-check/cinder-solidfire-iSCSI/c2ef099/ : FAILURE in 3h 02m 06s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"9a7d5a1194ed68837e2b40290bfca2f17f972556","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-13 19:24:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/42/764942/3/check/iscsi-driver-nimble/b9798f1/b9798f1.html : FAILURE in 1h 18m 39s","accounts_in_message":[],"_revision_number":3},{"id":"0146927f11cc30aad99a1fba3c414037b62ea07b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-13 20:48:39.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/764942/3/check/tempest-dsvm-full-bionic-py3/d82f42e/ : SUCCESS in 1h 55m 12s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/42/764942/3/check/devstack-plugin-ceph-dsvm-bionic-py3/c3a4581/ : FAILURE in 1h 12m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f2f677b856b0f1fef30fa7c5b39fef166e0e63aa","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-14 03:02:55.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/764942/3/20210414104346/NEC-ISCSI : FAILURE in 1h 18m 33s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/764942/3/20210414104346/NEC-FC : FAILURE in 1h 19m 07s","accounts_in_message":[],"_revision_number":3},{"id":"19fc3c6b133a94310d249d7e791f4875d25b63f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-03 14:29:48.000000000","message":"Abandoned\n\nStein transitioned to End of Life by change Icf9a539a7b8b and is accepting no more changes.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bd0c233b4c2833023ceec981fefa28c9f256cee2","revisions":{"17ab8a438a4a19c6f4ab0a689163e8ad6db57543":{"kind":"REWORK","_number":1,"created":"2020-12-01 14:12:12.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/42/764942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/764942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/764942/1"}}},"commit":{"parents":[{"commit":"13d48f8589d8981cebac6978f4b22d840f5c9132","subject":"Merge \"NetApp SolidFire: Fix replication\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13d48f8589d8981cebac6978f4b22d840f5c9132"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-01 14:20:51.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-12-01 14:11:56.000000000","tz":-180},"subject":"NetApp SolidFire: Fix clone and request timeout issues","message":"NetApp SolidFire: Fix clone and request timeout issues\n\nUsers are experiencing timeout issues in certain environments, mostly\nwhen volumes are too big (ie. multi-terabyte volumes), due to poor\nnetwork performance or upgrade issues that revolve around the SolidFire\ncluster. A viable solution is to make driver timeout values\nconfigurable in cinder.conf, so users can set these timeouts\naccording to their needs.\n\nThis patch adds two timeout settings to the SolidFire driver\n(for cloning operation and globally to all api requests), to\nallow users to set the appropriate timeouts for their environment.\n\nCloses-Bug: #1898587\nChange-Id: Ie330c76a5db0ea76d4fed5a6ae7b8736dadc8591\n(cherry picked from commit c607a82a992ebe63b4ae005746f2b9e8104bb3a0)\n(cherry picked from commit 3b2e7d662b6cca54d13d46161560a497d4ec5bd0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/17ab8a438a4a19c6f4ab0a689163e8ad6db57543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/17ab8a438a4a19c6f4ab0a689163e8ad6db57543"}]},"branch":"refs/heads/stable/stein"},"2fc981d593f713acef332ac95060ced4630584c3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-13 13:06:05.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/42/764942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/764942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/764942/2"}}},"commit":{"parents":[{"commit":"13d48f8589d8981cebac6978f4b22d840f5c9132","subject":"Merge \"NetApp SolidFire: Fix replication\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13d48f8589d8981cebac6978f4b22d840f5c9132"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-01 14:20:51.000000000","tz":-180},"committer":{"name":"Fernando Ferraz Silva","email":"sfernand@netapp.com","date":"2021-04-13 13:05:52.000000000","tz":0},"subject":"NetApp SolidFire: Fix clone and request timeout issues","message":"NetApp SolidFire: Fix clone and request timeout issues\n\nUsers are experiencing timeout issues in certain environments, mostly\nwhen volumes are too big (ie. multi-terabyte volumes), due to poor\nnetwork performance or upgrade issues that revolve around the SolidFire\ncluster. A viable solution is to make driver timeout values\nconfigurable in cinder.conf, so users can set these timeouts\naccording to their needs.\n\nThis patch adds two timeout settings to the SolidFire driver\n(for cloning operation and globally to all api requests), to\nallow users to set the appropriate timeouts for their environment.\n\nCloses-Bug: #1898587\nChange-Id: Ie330c76a5db0ea76d4fed5a6ae7b8736dadc8591\n(cherry picked from commit c607a82a992ebe63b4ae005746f2b9e8104bb3a0)\n(cherry picked from commit 3b2e7d662b6cca54d13d46161560a497d4ec5bd0)\n(cherry picked from commit b790e81e744755964ffa58c057c658a1b0d74581)\n(cherry picked from commit ddb88caad877f9e4dbd719336a87b6d7af327bc9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2fc981d593f713acef332ac95060ced4630584c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2fc981d593f713acef332ac95060ced4630584c3"}]},"branch":"refs/heads/stable/stein","description":"Edit commit message"},"bd0c233b4c2833023ceec981fefa28c9f256cee2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-13 13:10:19.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/42/764942/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/764942/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/764942/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/764942/3"}}},"commit":{"parents":[{"commit":"516852f2d67b3f82dd63658b16c15607681204cc","subject":"[stable-only] Set IPv6 params to fix tempest-slow-py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/516852f2d67b3f82dd63658b16c15607681204cc"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2020-10-01 14:20:51.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2021-04-13 13:09:05.000000000","tz":0},"subject":"NetApp SolidFire: Fix clone and request timeout issues","message":"NetApp SolidFire: Fix clone and request timeout issues\n\nUsers are experiencing timeout issues in certain environments, mostly\nwhen volumes are too big (ie. multi-terabyte volumes), due to poor\nnetwork performance or upgrade issues that revolve around the SolidFire\ncluster. A viable solution is to make driver timeout values\nconfigurable in cinder.conf, so users can set these timeouts\naccording to their needs.\n\nThis patch adds two timeout settings to the SolidFire driver\n(for cloning operation and globally to all api requests), to\nallow users to set the appropriate timeouts for their environment.\n\nCloses-Bug: #1898587\nChange-Id: Ie330c76a5db0ea76d4fed5a6ae7b8736dadc8591\n(cherry picked from commit c607a82a992ebe63b4ae005746f2b9e8104bb3a0)\n(cherry picked from commit 3b2e7d662b6cca54d13d46161560a497d4ec5bd0)\n(cherry picked from commit b790e81e744755964ffa58c057c658a1b0d74581)\n(cherry picked from commit ddb88caad877f9e4dbd719336a87b6d7af327bc9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd0c233b4c2833023ceec981fefa28c9f256cee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd0c233b4c2833023ceec981fefa28c9f256cee2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
