)]}'
{"id":"openstack%2Fcinder~766856","triplet_id":"openstack%2Fcinder~master~I5e35de28007558267f9215cbac03a38d7c74612c","project":"openstack/cinder","branch":"master","topic":"cinder-backup","hashtags":[],"change_id":"I5e35de28007558267f9215cbac03a38d7c74612c","subject":"Add support for RBD fast-diff feature for backups stored in Ceph","status":"MERGED","created":"2020-12-13 14:32:17.000000000","updated":"2021-03-24 20:42:47.000000000","submitted":"2021-03-24 20:39:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"766856-cinder-backup","meta_rev_id":"7ceda1af60ef146c7e54c634d54c6af623ce2bfc","_number":766856,"virtual_id_number":766856,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-24 20:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-17 20:18:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-16 15:18:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2021-03-17 04:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-16 14:14:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2021-03-24 17:53:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2021-03-24 17:53:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-17 20:18:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-13 14:43:59.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"},{"updated":"2020-12-13 15:06:47.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-13 15:25:34.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-13 15:52:30.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-13 15:57:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-13 16:07:06.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-13 16:16:46.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-13 16:35:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-13 16:43:27.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-13 17:28:03.000000000","updated_by":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"reviewer":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"state":"CC"},{"updated":"2020-12-13 17:44:15.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-13 18:37:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 12:58:27.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2020-12-14 13:38:30.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"},{"updated":"2020-12-16 08:40:44.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},"state":"REVIEWER"},{"updated":"2020-12-16 08:40:48.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REMOVED"},{"updated":"2020-12-16 14:45:05.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2021-01-13 09:49:14.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-01-13 09:49:14.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2021-01-13 09:49:49.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-13 17:32:13.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-01-14 17:40:28.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-16 12:46:18.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-21 21:10:06.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-01-22 14:33:01.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-02-18 13:47:15.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2021-02-18 23:17:08.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-02-18 23:41:40.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-02-19 00:29:50.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-02-19 03:42:59.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 05:28:57.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 13:26:03.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-02-20 04:37:31.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-02-20 08:11:41.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-02-20 19:55:31.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-02-23 21:08:52.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-25 11:30:47.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-25 12:36:51.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-02-27 03:58:50.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-27 06:53:57.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-28 14:57:32.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-03-17 04:55:12.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-03-22 14:57:05.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"514b33956b6353986c11c85bcef601894e450586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2020-12-13 14:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93006acf5e65b7132a268196a65f25e5aad1ec28","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2020-12-13 14:34:38.000000000","message":"Topic set to cinder-backup","accounts_in_message":[],"_revision_number":1},{"id":"e8ffe23a80888c18e2557f9438541e98dd145977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2020-12-13 14:44:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f2ebdfbe3e2fe0cc6aebdce6c630a0cd0a787638","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-13 15:25:34.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/2/check-not-vote/ext-cinder-zuul/a54eee9 : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":2},{"id":"0690721c0d833c825d7aa3f463bfd2f2a2f1d4d6","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-13 15:52:30.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/766856/2/iscsi/ : SUCCESS in 1h 07m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/766856/2/smb/ : FAILURE in 45m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1492ad4bda5d791465a9167b203f8293469f4946","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-13 15:57:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/56/766856/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/85c617c/85c617c.html : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3bb0e735dc50c81239b52a506b211aa8d0a154d9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-13 16:07:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/2/check/dsvm-tempest-cisco-zonemanager-job/7566526/ : SUCCESS in 1h 17m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e75ab2476ef44b4e52a70f9b5b440d828868f0c2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-13 16:16:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/56/766856/2/ : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":2},{"id":"43c9541e1fb9ba4c46edbdb0ba3928166f8970ab","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-13 16:35:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/2/check/DellEMC_Unity_FC/739fbd0/DellEMC_Unity_FC/295 : FAILURE in 1h 47m 44s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/2/check/DellEMC_Unity_iSCSI/f8aafda/DellEMC_Unity_iSCSI/280 : FAILURE in 28m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"af0e7206a0ff1b60e18369ed227008bf10348d89","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-13 16:43:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/2/check/DellEMC_XtremIO_FC/3df349f/DellEMC_XtremIO_FC/307 : FAILURE in 1h 41m 32s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/2/check/DellEMC_XtremIO_iSCSI/fe7d4b1/DellEMC_XtremIO_iSCSI/306 : SUCCESS in 1h 58m 17s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"79d425e584d3536903632cc113b2e569bc5678c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 16:46:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a658fa1be3a44bfea2304b5fd585f68f : FAILURE in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed0a36c1d95e43168d999b83c315e43a : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/012ecb52a0c24faabbe340b0f64217c7 : FAILURE in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d1b6e194cb0473d9125715d4a062ce6 : FAILURE in 12m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/093c230d88fb47a3892d1589ea7b866c : FAILURE in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1afdccdc1f3f4ca99b9b29565784d187 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/f46d06d19d5646108dea081796f2c50b : SUCCESS in 51m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8cb5dc88200c472ebd3407981c88777f : SUCCESS in 1h 08m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b7bb7dbe0bcb482e904feb8dfabcbf8e : FAILURE in 13m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89ed2dea81f2428e8fe68d33860e4398 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1818209cb7be4568a406bd5c27a752ed : SUCCESS in 6m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/185047a4651d42feae7626a5b141b8ab : SUCCESS in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d5e63285a78449b088a1ca40b0e559bd : SUCCESS in 39m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c550845b9f5d43ca892b4c258ed20e56 : FAILURE in 4m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a882fdb3e0064e3f8c2256c886e8a127 : SUCCESS in 1h 10m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e12f7018ff684dd4be5f4c6dfeeaf828 : FAILURE in 1h 19m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9805b94506834765ac184a1d3a21d84a : SUCCESS in 1h 21m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/076fcf1a6c7a46858d0e1870452ad33e : SUCCESS in 1h 05m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5aae84274bd415186ae396ab12b220e : SUCCESS in 1h 14m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3f0d1e1a81f44089a7f75c89c7e45b40 : SUCCESS in 1h 15m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/09079b46eaa94fe0b7807083191bfbdf : SUCCESS in 1h 57m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d201648c5ab497fab3f1ca96c5f25c3 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":2},{"id":"27ce6ac8d92526193ed064db7b00e8d79de51c72","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2020-12-13 17:28:03.000000000","message":"Patch Set 2:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":2},{"id":"76bac00348e25472710403d97e1bcbb669662ff5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-13 17:44:15.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-1/ns5_nfs : SUCCESS in 3h 11m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-1/ns5_iscsi : SUCCESS in 2h 52m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"e49aa97b9f1e4d1958ce93f3280bf9a20f1720b3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-13 17:47:13.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-2/ns5_nfs : SUCCESS in 2h 59m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-2/ns5_iscsi : SUCCESS in 2h 54m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"ddd6b60ef87cc0c173b38bf0a0f62664ea69a632","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-13 18:37:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/1/20201213233221/NEC-ISCSI : SUCCESS in 2h 26m 33s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/1/20201213233221/NEC-FC : SUCCESS in 4h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"589eee5749b604b416297431417c1bb0766b4123","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2020-12-13 19:39:53.000000000","message":"Patch Set 2:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":2},{"id":"0d3c7f2e7c6f35e5bd10242227e837d330578938","author":{"_account_id":19755,"name":"Shlomi Avihou","email":"shlomi@zadarastorage.com","username":"savihou"},"date":"2020-12-13 20:24:26.000000000","message":"Patch Set 2:\n\nrun-ZadaraStorage VPSA CI","accounts_in_message":[],"_revision_number":2},{"id":"e047d759aeae9cfc74fdd0bde1539587ebf0d256","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-13 20:51:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/2/20201214033727/NEC-FC : SUCCESS in 2h 02m 34s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/2/20201214033727/NEC-ISCSI : SUCCESS in 2h 13m 55s","accounts_in_message":[],"_revision_number":2},{"id":"ca484b7e51bd4189623573d7ecb3d7f52070469c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2020-12-16 08:40:48.000000000","message":"Removed reviewer Jenkins.","accounts_in_message":[],"_revision_number":2},{"id":"ef9b2aa4e240bf9461a724a3745ac15c88243ccf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-13 09:49:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fabcfa463603707c3d6d71c2d567c9a568044f5d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-13 09:49:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9cb6067c32c3d65be15f3bbe360d3e74aef5ae51","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-13 11:26:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/766856/2/iscsi/ : FAILURE in 1h 36m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/766856/2/smb/ : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":2},{"id":"75684cf506f80a481c4017c01f8d6f9c43808e31","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-13 12:24:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/2/check/dsvm-tempest-cisco-zonemanager-job/6296853/ : SUCCESS in 1h 18m 53s","accounts_in_message":[],"_revision_number":2},{"id":"dacda3376692aa295512fdc52fe86e7e3160665b","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-01-13 17:32:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nMissing release notes: https://docs.openstack.org/cinder/latest/contributor/releasenotes.html","accounts_in_message":[],"_revision_number":2},{"id":"b96ad41e72fc13e3d6c16f87eca2fcde74abb306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 23:25:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fa8d579042546799eda887e88f83c47 : FAILURE in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833c6f291b374a60b54d517492eeeb12 : SUCCESS in 9m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d23ac40272f40b09fac7b38bcebdaad : FAILURE in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/775b679f8e234494ac2dc1f6bf5474f8 : FAILURE in 11m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff665ecd9e31485286aaa80929f83d66 : FAILURE in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e4a17083f0b426d894088113c640bbf : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/5f2b14acc0d448ee9a2f5766e5b1c24d : SUCCESS in 56m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bedf0d16e6047c98ed7af8231a8f261 : SUCCESS in 1h 26m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/21d4f1374b6c4da995a321147ee36b07 : FAILURE in 10m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/db684cd6bfe542e0aa6746c05de1c0b7 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ca11beebedb4165955616d23562a3e4 : SUCCESS in 6m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f6699ffd21a14d20adb70f56c6caf3b3 : SUCCESS in 7m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/99bd9d5555e94606b0c53a57b52a8738 : SUCCESS in 40m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/30abf951db2b49ff923887edbab85d56 : FAILURE in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4766623cd2b7440c83f992b4075ec7e8 : SUCCESS in 1h 10m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5df5bfafe6264248bc2f9fbac673cc6f : FAILURE in 1h 13m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1b44c9197e29464aafa19b7c492d23f4 : FAILURE in 1h 17m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7cc561b8f8e54d639071e9fa159c6272 : SUCCESS in 52m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/272c21d8c52e45bcb7c6e07e5f77b832 : SUCCESS in 1h 32m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e16878fb3e4847ac94de87f478acfa94 : SUCCESS in 1h 24m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e95459a35db42aab59b4cde9f192c9a : SUCCESS in 2h 25m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/005a8976b4224ecd9c1105642d939a06 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":2},{"id":"83b8d6e2aed794fe10bb99a12bd6a43879cdb29e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-14 13:32:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48b14833b8f6465644ab9024d6e7187b3fa33d25","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-14 13:33:50.000000000","message":"Patch Set 3:\n\n(2 comments)\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Missing release notes: https://docs.openstack.org/cinder/latest/contributor/releasenotes.html\n\nSorry about that - added them now (patch set 3).","accounts_in_message":[],"_revision_number":3},{"id":"277b3f6b3356d274b57c97634758e1a5e6f400d7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-14 14:11:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/6a583d799c294cbb907de631f9a7fe64 : FAILURE in 14m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e18243649e613fcd82257c79f5d04e584f0b6df0","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-14 14:34:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/766856/3/iscsi/ : SUCCESS in 1h 02m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/766856/3/smb/ : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":3},{"id":"c8ef6dec08b921468605255eb63a9318691fce74","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-14 14:54:58.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/56/766856/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/24661ae/24661ae.html : FAILURE in 1h 21m 21s","accounts_in_message":[],"_revision_number":3},{"id":"fb8bedbcca6d287094755a4f03d84e9f24aa895d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-14 15:57:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/3/check/dsvm-tempest-cisco-zonemanager-job/8ce0d3d/ : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":3},{"id":"4b33a2d3441f658983679152873431ce79a97301","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-14 16:05:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/56/766856/3/check/dsvm-tempest-zadara-driver/03e1624 : SUCCESS in 1h 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed8dc6c993eb0a089d60bcee5ed2e2a20db88235","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-14 16:50:59.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-3/ns5_nfs : SUCCESS in 3h 18m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-3/ns5_iscsi : SUCCESS in 3h 13m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"5b8832574e9e69c88b19f611fc73c97969ddeeb3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-14 17:40:28.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/3/check/tempest-dsvm-full-focal-py3/b9ea526/ : FAILURE in 1h 47m 58s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/3/check/devstack-plugin-ceph-dsvm-focal-py3/d4b71a4/ : FAILURE in 2h 21m 28s","accounts_in_message":[],"_revision_number":3},{"id":"9f5a9e2cf11284aa06b73a3234753a978631c7f5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-15 00:44:34.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/3/check/DellEMC_XtremIO_FC/464a0e5/DellEMC_XtremIO_FC/514 : SUCCESS in 2h 12m 22s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/3/check/DellEMC_XtremIO_iSCSI/291d7cc/DellEMC_XtremIO_iSCSI/504 : SUCCESS in 1h 32m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"b485bdffeec73f34a490fec91671cfd21ae3959d","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-15 03:47:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/3/20210115103330/NEC-FC : SUCCESS in 2h 13m 11s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/3/20210115103330/NEC-ISCSI : SUCCESS in 2h 14m 26s","accounts_in_message":[],"_revision_number":3},{"id":"3a2f5709907fb4c95204055094ecaf0100809850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:12:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/707099cbf47544dc89415550871fe30e : FAILURE in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60285494cd2d4cea82cbe3801215960d : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ef47b2ebe0a4cab83f38ec7503ecaa2 : FAILURE in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edfe1335815848cea8b5094c73316db6 : FAILURE in 9m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36f865bf46c94c28be20e4ddee5815ab : FAILURE in 10m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae90d265b1345149f3f3bef1efa9bc8 : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/dfda84926bcb41c1b0b36d8dfebe51d1 : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/398bc093a65d43c3a1dfb6e89e18c039 : SUCCESS in 1h 09m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31972d763d5f441e902833cb1a5fd3b4 : SUCCESS in 6m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5e02c150ed8d472cb3979daab3554042 : FAILURE in 10m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/035485616d57456a85f3508ea28818c0 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/087d77e635a34d5589b496cc89f58687 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39c0de227be54fae9a36b9c863ea965d : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8ddcacdd93af4905aee751fad9e46c59 : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b7a5f11a2e374233a6bdad20148826c4 : FAILURE in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b4e61703420843fca92de5cbf3e098b3 : SUCCESS in 1h 01m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/10dde87daae443d1b935bf608ce60664 : FAILURE in 1h 20m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af19cce6c6044995a54d32bbb1125c4d : SUCCESS in 1h 44m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fbc99e978fe645a894615be12dd85482 : SUCCESS in 1h 11m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8a12ce27df8c48d7875c2d9c1e3f76ea : SUCCESS in 1h 24m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b6024d6fb65845769079f414acdfffd3 : SUCCESS in 1h 19m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b92cb37adaed4b39b03e4883a1c2f4cf : FAILURE in 2h 16m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac5b5bc3671e40268c350f6e5c948eee : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":3},{"id":"698f45440bb9860b18915e18ed0a6eb03376e482","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-01-15 11:17:09.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a2784d1457b63814f4369d198f358ca752efa907","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-15 15:13:25.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/3/check-not-vote/ext-cinder-zuul/7b8a2c8 : SUCCESS in 44m 39s","accounts_in_message":[],"_revision_number":3},{"id":"dea571bc54b09c354e4e90fae510fbc86110fb15","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-16 12:46:18.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/56/766856/3/check/DellEMC_PowerMAX_FC/ad4d159/DellEMC_PowerMAX_FC/391 : SUCCESS in 5h 21m 23s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/56/766856/3/check/DellEMC_PowerMAX_iSCSI/0612b11/DellEMC_PowerMAX_iSCSI/356 : SUCCESS in 4h 44m 37s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"063b371ec1c3ef6931a830c6668ee4fc9303ceed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-17 16:49:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"68b4735e3fb2dcf0a95cfcf064e367c99b64d3de","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-17 16:50:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c9a1dfe8c232466e4996f6533ea9046d1c9b3c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-17 16:55:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0cf03ecea61c8de4525f22155a10731d99c8c158","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-17 17:07:06.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d67f8d19c6b7d268b7cf085f173029fc4d73c0c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-17 17:08:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/56/766856/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/de77a5a/de77a5a.html : FAILURE in 6m 41s","accounts_in_message":[],"_revision_number":5},{"id":"0d31ffb8e32543d11be731ade7762354dee44b5a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-17 17:09:24.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/5/check/iscsi-driver-primera/6564522/6564522.html : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":5},{"id":"a18b6963a6ea63c07a17bfcb7e66163e3c7bc984","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-17 17:18:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/a56d92da4c1f4d7d9adb81b4feb38f24 : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":5},{"id":"70a3895badcdefb65fdafcd042af70add2829871","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-17 17:34:52.000000000","message":"Patch Set 5:\n\nrun-HPE Storage Primera CI","accounts_in_message":[],"_revision_number":5},{"id":"4c49b4a6885eda9017571cd24cee0263a588eb32","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-17 17:35:06.000000000","message":"Patch Set 5:\n\nrun-HPE Storage CI","accounts_in_message":[],"_revision_number":5},{"id":"6d3d992339213593d43378a68c30730ba850c213","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-17 17:37:52.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/5/check-not-vote/ext-cinder-zuul/f91fa93 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":5},{"id":"eb878c386276cded21a1464c0ea045152b0ae063","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-17 18:07:14.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/56/766856/5/check/dsvm-tempest-zadara-driver/fff0981 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4c8160dfca3fcee4ad0752b3be31165ddc43d4c0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-17 18:21:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/5/check/dsvm-tempest-cisco-zonemanager-job/afc81e0/ : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":5},{"id":"2a1e2c091aa18a71f203e02b6ec669ac296f388d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-17 18:25:12.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/766856/5/iscsi/ : SUCCESS in 1h 02m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/766856/5/smb/ : FAILURE in 1h 28m 51s","accounts_in_message":[],"_revision_number":5},{"id":"d173508a1fa8f2b2c8f2f52d8f332a27a89815df","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-17 18:29:15.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/5/check/DellEMC_Unity_FC/b4807da/DellEMC_Unity_FC/463 : FAILURE in 12m 23s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/5/check/DellEMC_Unity_iSCSI/ad7ccd2/DellEMC_Unity_iSCSI/428 : SUCCESS in 1h 30m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"4455a0bc82f272666408afe87fcb6ff6ae157b57","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-17 18:34:43.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/5/check/tempest-dsvm-full-focal-py3/608bcfe/ : SUCCESS in 1h 25m 02s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/5/check/devstack-plugin-ceph-dsvm-focal-py3/927eee2/ : SUCCESS in 1h 31m 21s","accounts_in_message":[],"_revision_number":5},{"id":"f534e3753ff8f1ab2d1d54d5410a044e91dfdd97","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-17 18:49:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/56/766856/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a84d626/a84d626.html : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":5},{"id":"7b09e348f5ded7a6ce06c9b543a7d28f852d54f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 18:51:38.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/394086ef1bb94e95bff751fd0b3aefdb : FAILURE in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637c42c38b9343698bbd4e98a63c0dbb : FAILURE in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ac31d7a369b4d379d669df19f7ff6b0 : FAILURE in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e72efd7f66b456f970d4e2f8bd5573d : FAILURE in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b0faf33441b4c0b991bec40372c527a : FAILURE in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5123294e84248cda637d5b7859aeaff : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/db36cb00f2624f6b8d4bf33d119d6d24 : SUCCESS in 55m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c33aab579834416b7c6d994cf17a233 : FAILURE in 1h 01m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6b9786416fa4b75aad5fe76f176e398 : SUCCESS in 5m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bf2085643d21477399ee468b936a09d6 : FAILURE in 10m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f364ddf3b9754401bfe6a323954b45a8 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40b55d6013e34887abf5be2d0d70b8a0 : SUCCESS in 5m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ed6e865151b34d5bb81ae5aaadf41188 : SUCCESS in 5m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca465909777e4ed7985423357603edd1 : SUCCESS in 41m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c723c2d7aea4d9c8c13c9314c03669d : FAILURE in 5m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32b263f241124cfbb187bc871245191c : SUCCESS in 1h 06m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3fb40418f44747fbb49c452b4ccddbc6 : FAILURE in 1h 02m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9423c4cdfa564705a755a5f59d83d918 : SUCCESS in 1h 24m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6aefa1c353b8461cb37ec57b1e5d0c4e : SUCCESS in 49m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f923530498ce43e386aee214ab3be18c : SUCCESS in 1h 54m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0b8bef25c5014415899c795d67bd5295 : SUCCESS in 56m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8e75bb85ccca48f3b48e255182a0993e : SUCCESS in 1h 50m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1ac92810ad4466f9762e5cb3a1bb25e : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":5},{"id":"b35f3e3eccbda882aad9b67f45e4446eda08ee1c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-17 18:53:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/5/check/iscsi-driver-primera/90cf970/90cf970.html : SUCCESS in 1h 17m 12s","accounts_in_message":[],"_revision_number":5},{"id":"6bfa1d3aab6d1b06a9c3f2e9db1587b1a2e95569","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-17 19:02:42.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/5/check/DellEMC_XtremIO_FC/10ea678/DellEMC_XtremIO_FC/545 : SUCCESS in 2h 04m 14s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/5/check/DellEMC_XtremIO_iSCSI/4275213/DellEMC_XtremIO_iSCSI/536 : SUCCESS in 1h 30m 26s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"5c67a0bc613757af6f91a2d8d95853687ab8c18e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-17 19:09:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/4/20210118014945/NEC-ISCSI : SUCCESS in 2h 14m 23s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/4/20210118014945/NEC-FC : SUCCESS in 2h 19m 38s","accounts_in_message":[],"_revision_number":4},{"id":"4f11d86c48dcbcbfec712ddf9249585aea8cfbdd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-17 19:55:17.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-4/ns5_nfs : SUCCESS in 3h 4m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-4/ns5_iscsi : SUCCESS in 3h 1m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"db36f78b0ce71e89af13c5505299e5f86a643c8e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-17 20:28:18.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-5/ns5_nfs : SUCCESS in 3h 8m 3s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-5/ns5_iscsi : SUCCESS in 3h 4m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"28e920df55f1aa564f09f8e2d234a5bba0e5eb61","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-17 21:23:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/5/20210118040930/NEC-ISCSI : SUCCESS in 2h 08m 47s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/5/20210118040930/NEC-FC : SUCCESS in 2h 14m 08s","accounts_in_message":[],"_revision_number":5},{"id":"a6266d9885a0f486ca8df67a60e0600957e4d459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-18 12:20:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b7947a907196b18f2eccb3610c1ad4949ae3e43b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-18 12:29:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/6/check/iscsi-driver-primera/17f471a/17f471a.html : FAILURE in 7m 23s","accounts_in_message":[],"_revision_number":6},{"id":"bf954c4ef44c4c1d3e6f0b861955fda7c796e15c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-18 12:35:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/56/766856/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/6a3741f/6a3741f.html : FAILURE in 12m 51s","accounts_in_message":[],"_revision_number":6},{"id":"b6536357bec9342a7c700e722f194e5cbdc2eb0c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-18 12:44:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/9ee19c90a413414fb08891aeac4b5fac : FAILURE in 19m 31s","accounts_in_message":[],"_revision_number":6},{"id":"f59b69aa5fa11e462ff1981a55115ac4fea6c4eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-18 13:05:04.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/6/check-not-vote/ext-cinder-zuul/c5f8ed6 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":6},{"id":"c608a575faa620b633f8ff4ae7b4092c17f45701","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-18 13:57:20.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/766856/6/iscsi/ : FAILURE in 1h 36m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/766856/6/smb/ : FAILURE in 59m 53s","accounts_in_message":[],"_revision_number":6},{"id":"5f80c9b3e33975fa815f0b8d291dea284b5ef8c4","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-18 14:27:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/56/766856/6/check/dsvm-tempest-zadara-driver/e02797b : SUCCESS in 1h 16m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cd516ed51297e428b281852876487aaf619317db","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-18 15:00:15.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/6/check/tempest-dsvm-full-focal-py3/4fe8c22/ : SUCCESS in 1h 47m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/6/check/devstack-plugin-ceph-dsvm-focal-py3/17aa367/ : SUCCESS in 2h 01m 40s","accounts_in_message":[],"_revision_number":6},{"id":"8fbbde633f531fb858901d5d4fdb823c7e1dc114","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-18 15:01:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/6/check/dsvm-tempest-cisco-zonemanager-job/27d6ac2/ : SUCCESS in 1h 15m 38s","accounts_in_message":[],"_revision_number":6},{"id":"8acf6e54aa7294719ff74d12a1751027eae0662f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-18 15:38:15.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-6/ns5_nfs : SUCCESS in 3h 17m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-6/ns5_iscsi : SUCCESS in 3h 2m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"d17e1660892062f9232227877c4ecbaa21b6d371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:54:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ee048a2c5824c3abac8b60579e42278 : FAILURE in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0548fe2a35a34d1a9c4c6b305b1c9de9 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c25f8f8131524f47a446286f11213571 : FAILURE in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eee5fa079a564f0bae87ed22bf798fe2 : FAILURE in 10m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67deabe5ac2b4bc391b9718ae7a61923 : FAILURE in 11m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ba417c357d425c8f8ea298d4f52737 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/2227eaed57c64708b02e0983597b9d31 : SUCCESS in 58m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5088cac1b52b4f17bc237bcb53800da9 : SUCCESS in 1h 07m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15c10cb802664a38b51b1aeafadd5fb7 : SUCCESS in 7m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/831c7dd8e2c8414dbce6280a23be2dfa : FAILURE in 10m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/676138b0df764f5692a23ed4d51a0cc8 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27480f8f1046403d95dc899940433825 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/88a7d5de81c84a139ff14a383218861c : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ffdb432b2b904ae0bcf3e28347a8ef0e : SUCCESS in 59m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1570e2cde0ff47c2833ce0bdb54c1025 : FAILURE in 5m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/de48495ea1534ee684ff86949a2917e4 : SUCCESS in 1h 00m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/997c0437fbf04331aafb1c0122b59efb : FAILURE in 1h 19m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/750f30ba16b043c7a0941e115e3f070e : SUCCESS in 1h 28m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/594b7b1d9e754cf7a71d0a8df87ded80 : SUCCESS in 1h 11m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac5fbe0823634180ba43af97798a20f1 : SUCCESS in 1h 24m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/33b01d07008d4c54b814b3ac9b38c38d : SUCCESS in 1h 30m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86a04c302e7047c286182ef7000a592d : FAILURE in 2h 22m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14c28c79199e42409c5c24d21b885dd4 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":6},{"id":"eeef319fa730d9e216ed606acf2299ca2cf63ba9","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-18 18:13:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/6/20210119005341/NEC-ISCSI : SUCCESS in 2h 13m 33s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/6/20210119005341/NEC-FC : SUCCESS in 2h 20m 07s","accounts_in_message":[],"_revision_number":6},{"id":"5e6e7418b08568b8ad44c08cf77da69510335ad2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-18 18:57:26.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/6/check/DellEMC_Unity_FC/84c59a7/DellEMC_Unity_FC/468 : SUCCESS in 1h 30m 25s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/6/check/DellEMC_Unity_iSCSI/331db5b/DellEMC_Unity_iSCSI/433 : FAILURE in 1h 41m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"28878397007e30882ade43cbda8a3cd7cf5faa60","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-18 19:21:49.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/6/check/DellEMC_XtremIO_FC/b7ec7da/DellEMC_XtremIO_FC/550 : SUCCESS in 2h 07m 47s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/6/check/DellEMC_XtremIO_iSCSI/c371d17/DellEMC_XtremIO_iSCSI/541 : SUCCESS in 1h 32m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"58f7b8be9b96b41f540bbdba0ec6c57317461f28","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-18 19:25:02.000000000","message":"Patch Set 6:\n\nrun-HPE Storage Primera CI","accounts_in_message":[],"_revision_number":6},{"id":"b0a51933c46d241d9a641aaac0510cd9678e16b4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-18 19:26:24.000000000","message":"Patch Set 6:\n\nrun-HPE Storage CI","accounts_in_message":[],"_revision_number":6},{"id":"c768dd9d6fbe2a4227ec04b4dab06e327a2e6bb9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-18 19:33:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/6/check/iscsi-driver-primera/6fd450e/6fd450e.html : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":6},{"id":"059cd2c64d070e8ed8e5dd9839c7dca4c25e9504","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-18 19:53:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/56/766856/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4009756/4009756.html : FAILURE in 25m 48s","accounts_in_message":[],"_revision_number":6},{"id":"e4ed4d5137528037674366135f8b87cf26685ecb","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-01-21 21:10:06.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nHi Christian, thank you for your patch. \n\nComments inline. \n\nOn the one hand, I\u0027m not sure why cinder doesn\u0027t use FAST_DIFF maybe there\u0027s a reason why we don\u0027t enable this before so any input from core reviews is welcome. \n\nOn the other hand, it would be nice to add this configuration somewhere in the CI as a follow up.","accounts_in_message":[],"_revision_number":6},{"id":"0f5f2e48926ffe83e4d51ed77a509753a58433fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-22 13:58:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"866ba59e28e0b873d52be1b3afa6260149754b84","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-22 13:59:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"10d9d0d022945762ed66241a7d78f29b5ea0fa56","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-01-22 14:33:01.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* KIOXIA Kumoscale CI http://104.254.65.37/refs-changes-56-766856-7/1611323931\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":7},{"id":"c5d4beed5f67295df14bee9361530e78f051952d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-22 14:42:50.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/7/check-not-vote/ext-cinder-zuul/76b794f : SUCCESS in 43m 06s","accounts_in_message":[],"_revision_number":7},{"id":"40271ba1f136b94cc5c40a0e2a4844318cdadae5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-22 15:08:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/56/766856/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cf45a21/cf45a21.html : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":7},{"id":"a5a151254ffe0d80fea8c2b72d83d8d9c1da2a57","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-22 15:21:02.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/7/check/iscsi-driver-primera/faaae1b/faaae1b.html : SUCCESS in 1h 20m 39s","accounts_in_message":[],"_revision_number":7},{"id":"70a13ae7252e0b01a01aa35faff3dee51dd58f52","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-22 15:24:33.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/7/check/dsvm-tempest-cisco-zonemanager-job/554b424/ : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":7},{"id":"59a1638788bceb85300baf083cf528e841d2d9e5","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-22 15:26:31.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nThank you Sofia for diving into my code and reviewing it!","accounts_in_message":[],"_revision_number":7},{"id":"f2f8df17a458f8fba056e88f7ec031abc2d41bc8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-22 15:29:50.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/766856/7/iscsi/ : FAILURE in 1h 30m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/766856/7/smb/ : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":7},{"id":"dee8e94fcf62b02d2381c44944bd1baf9ab5ad1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 16:18:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0572aac795cd432ebcb7fff55ad5d314 : SUCCESS in 13m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c538fe3a5a14ba38b1f4ab38ac1f01a : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e01ae119e8da4635a7a0ce785305064e : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26523aa113194b209efa2dff32bb0d8a : SUCCESS in 10m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b85c034fe1c45409cc9b0561baa2fc8 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328ae37c21e74b758945561d59c0c33a : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/f02e3d5154974034a568485f57fb780a : SUCCESS in 1h 00m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/785acbdb1a6547f1ba099259d353078f : SUCCESS in 1h 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3a3ac5d70b849b3ba0d39d1f5d72b4e : SUCCESS in 6m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1b427029488a4899b04c7f6f45f8e31a : SUCCESS in 12m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b6f988feb0a42d68b4e32ac93134155 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8616ffc341624ecc8ed1fe6d9022aeb2 : SUCCESS in 7m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/69e7fadbec7d4903aeac5a070287e12f : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/464caf44c9da46e2a6ecaa801ec47cce : SUCCESS in 43m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58ffa7cecfc44ed195aab08b84f9b70e : FAILURE in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bb1e91517f024f5f98796073d440ce6f : SUCCESS in 1h 07m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bcbafc4b547e4f3e8efb4c5f5851dc61 : FAILURE in 1h 17m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e47bd43524047b28890caf9b665b237 : SUCCESS in 1h 30m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3b2175e935da4b2e873ae5a35936b5f1 : SUCCESS in 1h 08m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/668f4407f4924789a7d5337dc1e1582c : SUCCESS in 1h 36m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db013365d7a7465788d1dc636bb607d3 : SUCCESS in 1h 16m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/249bd477312a4e16ae619429162ad6b4 : SUCCESS in 2h 13m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ce0ee173d0c4ffc947d733819951cb6 : RETRY_LIMIT in 20m 38s","accounts_in_message":[],"_revision_number":7},{"id":"8322b0e3dcbbe2d88a0cc33d26eccec0bd853849","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-22 17:10:38.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-7/ns5_nfs : SUCCESS in 3h 11m 9s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-7/ns5_iscsi : SUCCESS in 2h 51m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"e320de68ce779b3bdfb159a0287eae3e29836042","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-22 17:55:13.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/7/check/tempest-dsvm-full-focal-py3/b8dd0d8/ : FAILURE in 1h 31m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/7/check/devstack-plugin-ceph-dsvm-focal-py3/2be1d34/ : FAILURE in 1h 45m 36s","accounts_in_message":[],"_revision_number":7},{"id":"6d9ec80dbd3f6301a2c71d552d315530890de0da","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-23 00:41:01.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/7/check/DellEMC_XtremIO_FC/809ed06/DellEMC_XtremIO_FC/601 : FAILURE in 10m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/7/check/DellEMC_XtremIO_iSCSI/1de4a56/DellEMC_XtremIO_iSCSI/595 : SUCCESS in 1h 44m 43s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"853d0c9c60e6cce1c14753edcb794e94eac26213","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-23 01:02:02.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/7/check/DellEMC_Unity_FC/d590fd3/DellEMC_Unity_FC/523 : SUCCESS in 1h 34m 25s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/7/check/DellEMC_Unity_iSCSI/7a6a57f/DellEMC_Unity_iSCSI/486 : FAILURE in 10m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"5f551e2a64563da3bc79ffcb4c8f424d6e881423","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-01-23 04:54:31.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"e03d8f3d8662a8a3613fdd856640fced4eccb62d","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-23 18:00:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/7/20210124002741/NEC-ISCSI : SUCCESS in 2h 27m 46s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/7/20210124002741/NEC-FC : SUCCESS in 2h 33m 04s","accounts_in_message":[],"_revision_number":7},{"id":"cc59cd7b58607ba7508f2f75d99183f756b69cfc","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-01-28 09:38:49.000000000","message":"Patch Set 7:\n\nSofia, Gorka\n\ncould this potentially get your +1 then?","accounts_in_message":[],"_revision_number":7},{"id":"bd06f0eadee275f76fddc8fd3d9ecd44bdbb3fb6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-18 12:21:15.000000000","message":"Patch Set 7:\n\nMaybe someone already asked for it, but could you try to create a DNM review where this options is enabled in devstack? We have at least one ceph-based job whose tempest tests may at least show that enabling the option doesn\u0027t break anything.","accounts_in_message":[],"_revision_number":7},{"id":"637ce4d5aecb024004f41e129864f0aa48afa0a9","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-18 12:27:51.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e Maybe someone already asked for it, but could you try to create a DNM review where this options is enabled in devstack? We have at least one ceph-based job whose tempest tests may at least show that enabling the option doesn\u0027t break anything.\n\nI don\u0027t quite understand what is required here. Could you kindly point me to an example or some documentation on what I would need to do please?","accounts_in_message":[],"_revision_number":7},{"id":"00017503134ddf0fc2cf2904af8743e416d75b82","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-18 13:47:15.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"863c354a6dd82c726cb2d0d7f72ebb7b699760fc","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-18 13:47:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"228324493d6f57cc75838b0936aa695ebd11684d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-18 15:11:42.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/766856/7/iscsi/ : FAILURE in 1h 19m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/766856/7/smb/ : FAILURE in 1h 23m 47s","accounts_in_message":[],"_revision_number":7},{"id":"43b9cd00a3bc423cc976449d0c8e5fd931d6e96f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-18 15:11:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/7/check/dsvm-tempest-cisco-zonemanager-job/8f8a268/ : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":7},{"id":"c2bdaf616f570f15685a6ca7931c7fb26e4b661e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-18 16:37:29.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/7/check-not-vote/ext-cinder-zuul/4e9a75f : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":7},{"id":"2e8f7790d38fdd6ce7ea00b5fe9691a70888979a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-18 19:30:56.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/7/check/tempest-dsvm-full-focal-py3/d9d9491/ : SUCCESS in 2h 20m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/7/check/devstack-plugin-ceph-dsvm-focal-py3/c1352a0/ : TIMED_OUT in 2h 40m 06s","accounts_in_message":[],"_revision_number":7},{"id":"4416f58a3e915ce8a2aa6a4a9781de3b9fc27759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 22:20:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d295ed6caa47899876ab6b260d67a2 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/692ca7f18bc34440ba25991ed9b40434 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a314c535c6440aba5911fae0e6eb69b : SUCCESS in 11m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ff9cab00f1b4ef48f8a2852db759fc8 : SUCCESS in 10m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5f17551317a487da55b17d9a04cf2c0 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de840bd7d3e438fb3141bd6db776755 : SUCCESS in 13m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/8fd34955c88c4cbeb468abd5287d2bcf : SUCCESS in 1h 01m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/138c0aa0cdb741689918a19053326609 : SUCCESS in 1h 06m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05aa97eab7644622b17ba95adc381332 : SUCCESS in 6m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a5b9fc68155a4ca7964d7accdfe11fc6 : SUCCESS in 13m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee6a0763ec064a29a54ee3a8bea8c59f : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4dc903229f4f48ee94f264ba793a9bfd : SUCCESS in 6m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eda7049d967e46a2a6f5bdbb61ef321c : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a452a0f06db54398bc1c9cc9ecd4774b : SUCCESS in 45m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f655963ff9c4d5db9136307a5adaef5 : FAILURE in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc0f87bf95f149bd9f19c0958da0828d : SUCCESS in 1h 17m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6d63c0155a1c49aa94777703e7f8c494 : FAILURE in 1h 18m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ac0e583e192740a38e00e33f6cf8dd86 : SUCCESS in 1h 29m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/725d7a951d93497ba1aa6296e332a67a : SUCCESS in 1h 11m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/478baf857b6344fb9620630a45dbb32a : SUCCESS in 1h 23m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/286a179359d143adbb91209fda1f557f : SUCCESS in 1h 22m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/137444956b2e4d1f98e6d22d492b24f4 : SUCCESS in 2h 04m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/438f4e71f314452bb1329163e1b159a2 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":7},{"id":"a50ae6c9d7a21941717ee03702a1dbee5782afa8","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-18 22:46:45.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-56-766856-7/1613686810 : FAILURE in 26m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":7},{"id":"6f6c1ab038df2d3dab1e3ea4cd4b18bd4f584a41","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-18 23:17:08.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/56/766856/7/check/DellEMC_PowerStore_iSCSI/a9c68ba/DellEMC_PowerStore_iSCSI/443 : SUCCESS in 56m 26s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/56/766856/7/check/DellEMC_PowerStore_FC/69b4dae/DellEMC_PowerStore_FC/437 : SUCCESS in 47m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"82bee7202ff6cd89097e891bc07e191d89bbba93","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-18 23:41:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/56/766856/7/upstream-solidfire-check/cinder-solidfire-iSCSI/9d5c65d/ : SUCCESS in 1h 19m 53s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"dacb5d34d1bbdb8d7291868f085ea662b8a92ff9","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-19 00:29:50.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/56/766856/7/check/DellEMC_SC_FC/beb222a/DellEMC_SC_FC/360 : FAILURE in 1h 59m 09s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/56/766856/7/check/DellEMC_SC_iSCSI/66df9de/DellEMC_SC_iSCSI/358 : FAILURE in 2h 09m 06s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"3e93290969644749c4f41f37de7c8551c5ada76e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 03:42:59.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/56/766856/7/check/DellEMC_PowerFlex/568abf7/DellEMC_PowerFlex/411 : FAILURE in 2h 40m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"36b4cd22d206bd83abbdaa2f60c47073e032b1cb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 05:28:57.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/56/766856/7/check/DellEMC_VNX_iSCSI/8285734/DellEMC_VNX_iSCSI/413 : FAILURE in 1h 21m 02s\n- DellEMC_VNX_FC http://publiclogs.emc.com/56/766856/7/check/DellEMC_VNX_FC/24649e5/DellEMC_VNX_FC/432 : FAILURE in 56m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"8c5fc86d9f94c6bd94c9fc359f88b82304b4097f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-19 13:26:03.000000000","message":"Patch Set 7: Code-Review+2\n\nThis kind-of seems to be a feature listed as a bug but it is a small change so I am willing to consider it.","accounts_in_message":[],"_revision_number":7},{"id":"9df821d616d2a4d0b9b5d392fbf3b4b7a05bb7dc","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-19 15:40:59.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7: Code-Review+2\n\u003e \n\u003e This kind-of seems to be a feature listed as a bug but it is a small change so I am willing to consider it.\n\nThanks a 1000 Jay. Is this all that is required for me to do or do I need to find more reviewers?\n\nCould you please explain what I should have done differently?\nNot open a bug in launchpad and just produce a patch? Certainly there are specs and blueprints ... but this is just a simple option switch. So producing such documents seems excessive.","accounts_in_message":[],"_revision_number":7},{"id":"dae10eef05d43a3cad1fe07884fed4a27680c71b","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-02-20 04:37:31.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/766856-7/Seagate-iSCSI : SUCCESS in 45:54.06\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/766856-7/Seagate-FC : SUCCESS in 40:01.38","accounts_in_message":[],"_revision_number":7},{"id":"31b2e073f952f77532b61437d215a94b529ed1dc","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-20 08:11:41.000000000","message":"Patch Set 7:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/766856-7/Lenovo-iSCSI : SUCCESS in 50:24.68\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/766856-7/Lenovo-FC : SUCCESS in 46:22.24","accounts_in_message":[],"_revision_number":7},{"id":"270f8a68f2706df73c5fba8c04ec6242f3f8aa38","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-20 09:21:47.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/7/check/DellEMC_XtremIO_FC/9c2409e/DellEMC_XtremIO_FC/809 : SUCCESS in 1h 57m 54s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/7/check/DellEMC_XtremIO_iSCSI/5a85bd3/DellEMC_XtremIO_iSCSI/811 : SUCCESS in 1h 40m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"8367808c5566ab4eead24a88ce49c106030e44cb","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-20 19:55:31.000000000","message":"Patch Set 7:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/766856-7/HPMSA-iSCSI : FAILURE in 1:06:15\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/766856-7/HPMSA-FC : SUCCESS in 45:40.99","accounts_in_message":[],"_revision_number":7},{"id":"98351c65e6ecb24a194e083bc989291289ff9907","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-23 21:08:52.000000000","message":"Patch Set 7: Code-Review-2\n\nJust putting a hold on this until tomorrow.  We\u0027ll discuss at the weekly cinder meeting.\n\nThis came up during the festival of reviews (I think), the issue being that instead of adding a config option, we should just go ahead and do it if the feature is available, and fall back to the current behavior otherwise.  I\u0027ll update this review after tomorrow\u0027s meeting.","accounts_in_message":[],"_revision_number":7},{"id":"d78be1ecfb4425a8892a4d35158d2bc4e5efcf8e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-24 16:28:03.000000000","message":"Patch Set 7:\n\nWe discussed this at today\u0027s Cinder weekly meeting:\nhttp://eavesdrop.openstack.org/meetings/cinder/2021/cinder.2021-02-24-14.00.log.html#l-138\n\nThe consensus is that we don\u0027t want this to be controlled by a cinder configuration option.  Please revise the patch to just use fast diff if it\u0027s available.\n\nThanks!","accounts_in_message":[],"_revision_number":7},{"id":"2251dfb5c4d6b5689836dde44dbbc2a65a5a0576","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-24 16:36:25.000000000","message":"Patch Set 7:\n\nAlso, as you are going to change it, can you please mention it\u0027s about \"backup\" in the first line of the commit message (and not about any normal volume)?","accounts_in_message":[],"_revision_number":7},{"id":"464bfc8f8b4cc164e75d41ffb9070f86d486bb56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-25 08:18:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"48ee4d666941d54deeeaa027f68a83fcff868c40","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-25 08:19:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2d53e965ea6d5a38e6e3c30079c840a2d949bc1f","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-25 08:20:54.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 7:\n\u003e \n\u003e We discussed this at today\u0027s Cinder weekly meeting:\n\u003e http://eavesdrop.openstack.org/meetings/cinder/2021/cinder.2021-02-24-14.00.log.html#l-138\n\u003e \n\u003e The consensus is that we don\u0027t want this to be controlled by a cinder configuration option.  Please revise the patch to just use fast diff if it\u0027s available.\n\u003e \n\u003e Thanks!\n\nDid upload a revised patchset 8.","accounts_in_message":[],"_revision_number":8},{"id":"3f57e50253e35b332b15d941bbb4e192f15c7fb9","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-25 08:25:36.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 7:\n\u003e \n\u003e Also, as you are going to change it, can you please mention it\u0027s about \"backup\" in the first line of the commit message (and not about any normal volume)?\n\nDone.","accounts_in_message":[],"_revision_number":8},{"id":"f107f88aa1531b2dcdf4120c4686bb61c895d7a6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-25 09:04:08.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/8/check-not-vote/ext-cinder-zuul/8cc02d3 : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":8},{"id":"194234059951aa1587a1de5552bd2db48005ba76","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-25 09:34:03.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/8/check/iscsi-driver-primera/d921d76/d921d76.html : FAILURE in 7m 25s","accounts_in_message":[],"_revision_number":8},{"id":"4b22e53d557dbfc362c12da28fcace75cbc22ee7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-25 10:40:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/56/766856/8/ : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":8},{"id":"960158ee6f0437b8c50e0b0e763418c059dc2d3b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-25 11:20:27.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/8/check/dsvm-tempest-cisco-zonemanager-job/af76e0f/ : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":8},{"id":"a09ba5f91b2478666118b26fae2c40ff73df4245","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-25 11:30:47.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/56/766856/8/check/dsvm-tempest-TOYOU-ACS5000-driver/6601918 : FAILURE in 58m 36s","accounts_in_message":[],"_revision_number":8},{"id":"6ca26ebcaac4fcaf1b685ce41d7ebf32ccc5be92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 12:16:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/353ae263b1954f7ab01abf19087f28ad : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7486a4168e74e988dc3815a8f3011e1 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94f19dbfe74e48699e85229fe073de05 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8659d02e72c0423799bb33b12f11880a : SUCCESS in 9m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9307ca55901c4fe6afcc08127a30baee : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca9b11305a44440ae0326981339f4ff : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/51743411672744b29c01457d9474dce7 : SUCCESS in 1h 01m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d8b25eb70cd4b6dbe16c002d0599452 : SUCCESS in 1h 22m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbed0742bfa54b759d7b3e1e105bb36a : SUCCESS in 6m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/79417c336e464410993789df5c14be04 : SUCCESS in 13m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3cc7a9452b54ab0b9b412c1357d5d48 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a179a78f76b430ea0455da4838f5e4c : SUCCESS in 9m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9edcd9630a8434ca8fb22356f7fd78f : SUCCESS in 9m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d0afc11365bc4084811e6c4c0dfef67c : SUCCESS in 54m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b44cdb9cd07c456092b464f3364ab976 : FAILURE in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f3f5673dad594774a7345d0a3d967ed5 : SUCCESS in 1h 19m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/22c4661557544be69689a2c8ba35cb76 : FAILURE in 1h 30m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/815e2269746f448b931eeae57d672892 : FAILURE in 1h 29m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5f6ff5c6cfb493bb7933f307150bbea : SUCCESS in 1h 09m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0292ed09ef764a6181fa3ea22c6e51bc : SUCCESS in 1h 22m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/63dbb5acb8634d77a52c56c59579cdcc : SUCCESS in 1h 24m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35ecb2cf2473423cad7b3b0a41563e04 : SUCCESS in 2h 18m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81d219e3fa88417fa6e440b49f6af7a6 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":8},{"id":"91176e256cf60aa8a4d599fb63aa629aebb2cf05","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-25 12:18:57.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-8/ns5_nfs : SUCCESS in 3h 18m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-8/ns5_iscsi : SUCCESS in 3h 11m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"cd4eaebf998a159f27d0ee3d8104023748625db1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-25 12:36:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/766856/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0e69e2f/ : FAILURE in 2h 26m 25s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/766856/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/64c911a/ : SUCCESS in 1h 50m 12s","accounts_in_message":[],"_revision_number":8},{"id":"541b12a9ff78fab79f0952f8c854c933c3290fb9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-25 14:04:24.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/8/check/DellEMC_XtremIO_FC/727d5b5/DellEMC_XtremIO_FC/864 : SUCCESS in 1h 56m 38s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/8/check/DellEMC_XtremIO_iSCSI/5a2e841/DellEMC_XtremIO_iSCSI/869 : FAILURE in 12m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"0865b26660f01124c4ee5d2c9736d9d8b84f47e6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-25 18:06:47.000000000","message":"Patch Set 8: -Code-Review\n\nDropping the hold on this patch.","accounts_in_message":[],"_revision_number":8},{"id":"c88d3e5cafb69417dbd990c83fab49278304b509","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-25 18:08:18.000000000","message":"Patch Set 8:\n\nrecheck cinder-tempest-plugin-lvm-lio-barbican backup test failures","accounts_in_message":[],"_revision_number":8},{"id":"39041de72a8c35f66e188b63697bc724ac7c4dd5","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-25 18:08:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"eb64de7e2d5abdf597985499dc49931447e71c83","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-25 18:10:49.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cd93452c82864b0495e84bfaa7104fa9 : POST_FAILURE in 1h 54m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7b728a04067d4e27b1f3ee28b8e1f87f : POST_FAILURE in 1h 45m 41s","accounts_in_message":[],"_revision_number":8},{"id":"6f383a86a24208e5f2d0b41151225b0acf890162","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-25 18:51:57.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/8/check-not-vote/ext-cinder-zuul/c548fef : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":8},{"id":"a16521572bb7395c95b45145ac53f8346f604c28","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-25 19:35:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/8/check/dsvm-tempest-cisco-zonemanager-job/935c32c/ : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":8},{"id":"8c1d8960d4f1455c2fe3b491843dec2f42588360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 22:43:13.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/432d91a8090c47a0abef1327d8519965 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aaa263f57314bf9a947db56a97c8b7e : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c548463a07d143e3a5281eb44b94d86c : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffa1f5da4ed24155a216d488daee57cc : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb8e7ad782fd414a8a68235366b28252 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deda349c3efc4530b9cfa866202166e3 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/3cc9663ce2244b4982f6ad0fbcb4e8fa : SUCCESS in 1h 04m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c16ce7917e0b4d80857ef0cc6e886b81 : SUCCESS in 1h 35m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22f1ce0b9d4148bca29ae8fc098ae6e2 : SUCCESS in 7m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/910b5eed4df44034a397168de9203590 : SUCCESS in 14m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/15a3a00587034deb83a7ac57a66361f0 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3021c933e9741f9adbe05c473697825 : SUCCESS in 6m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5d4dc756c4ee40eba4f7722ca9b7d000 : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f98ef0bc270c41a4af06af8627f52e1a : SUCCESS in 41m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2cf487500b6241a39ae4449a9611b1a6 : FAILURE in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9663238f8966456bb06aeadd34abad30 : POST_FAILURE in 1h 06m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f003c2c8416e46d7a1ccd1411584725b : FAILURE in 1h 29m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3ab929e2e771452492f73bf6cd901856 : FAILURE in 1h 28m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38f1f83b1f634f3480bbb05826b1d57c : SUCCESS in 1h 20m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b2fb64254ab64bb59d74aecd53681cb4 : SUCCESS in 1h 29m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d7017d8d64e74355a0246d3353fc50f8 : SUCCESS in 1h 31m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0ff785e3f56c45c49da1664d07051c64 : SUCCESS in 1h 47m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/645a602b74c848d6b5470bf4db6a9b4b : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":8},{"id":"68bb1cdce3b38243b1c6139af3f7b15d0766d58f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-26 20:18:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/8/20210227025420/NEC-ISCSI : SUCCESS in 2h 14m 56s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/8/20210227025420/NEC-FC : SUCCESS in 2h 23m 50s","accounts_in_message":[],"_revision_number":8},{"id":"72842e9c4ade7794657a645e6e44295db7046291","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-26 22:12:35.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"ab85b1b517918b4c2cbb354c00a72ac0ae51970c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-26 22:13:11.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fcb024e07b4932678568e3a07d5bcffeaad0a20b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-26 22:13:18.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"ecead27e904289bdffde67acef4eae7686605c23","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-26 22:13:40.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"f38ed7f776b935d5c1101f0ae2cc04a198328eed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-26 22:16:18.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"1b6b8e43f374de9a07ff8d77497779eb34835c2f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-26 22:26:05.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/56/766856/11/check/iscsi-driver-nimble/82babe6/82babe6.html : FAILURE in 7m 12s","accounts_in_message":[],"_revision_number":11},{"id":"c5c3367289a3b2dfd7b040cd4e24539635e081c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-26 22:52:50.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/11/check/iscsi-driver-primera/a3147ca/a3147ca.html : FAILURE in 34m 52s","accounts_in_message":[],"_revision_number":11},{"id":"c86341a3c697f8495af49e8393f0c60194b5da1f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-26 23:01:46.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/11/check-not-vote/ext-cinder-zuul/3b6aa04 : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":11},{"id":"669647bd8df1c3993e6a8e75e1dceebac15abea4","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-26 23:32:17.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/56/766856/11/check/dsvm-tempest-zadara-driver/6f640aa : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f37af631e7334d012c8a249fbe22252b2805ad25","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-26 23:39:50.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/11/check/dsvm-tempest-cisco-zonemanager-job/aee6290/ : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":11},{"id":"49d2f677c1d643b99cf8c68abf8515aa9d61d041","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-26 23:41:19.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/56/766856/11/ : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":11},{"id":"6c10f6d8d04c9a90b6feaf84a54075e6d372a95a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-27 00:02:05.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/766856/11/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/482b441/ : SUCCESS in 1h 44m 20s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/766856/11/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0b73331/ : SUCCESS in 1h 27m 44s","accounts_in_message":[],"_revision_number":11},{"id":"f89f57d67c336e0fdb841d96ce8754a0320a7991","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-27 00:14:57.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1ba86200e9de4a75aff37284bfe5b902 : POST_FAILURE in 1h 53m 36s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/3e97a167d2f741c7a3cf8691c85c1276 : POST_FAILURE in 1h 52m 53s","accounts_in_message":[],"_revision_number":11},{"id":"38d868c725d589e1e6f2ef494d63a80c4a4932bc","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-27 00:17:33.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/56/766856/11/check/dsvm-tempest-TOYOU-ACS5000-driver/37cc877 : FAILURE in 1h 00m 12s","accounts_in_message":[],"_revision_number":11},{"id":"f8a707771a91dd969232f8ad2ae58b5a0341dfc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 00:53:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0220da0e4f749caac551604fb4f61fd : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e0c552ae684398a6011691799db097 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78f35c21eef949fc803df09a95b3f376 : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/494dcc8857ef498197eb001d58a020b8 : SUCCESS in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf9051bc7e94478683246c990c31d32f : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e529c52d13f3484d868d464efb1b15bd : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/c7aeca14fd5b4673b77d19845daa26c5 : SUCCESS in 1h 05m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f2393db31244f679eb0b2758a650219 : SUCCESS in 1h 39m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8183c92aa18a438597bdfca3d6edb1bb : SUCCESS in 6m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b02567ca505745ff9e1fbd7d7dc8c50b : SUCCESS in 17m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01536e68e2c64d419577e4aa8bc0ce8b : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b00be44732e4ea2bd91def36bf6e314 : SUCCESS in 8m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68b15f757d014601885e3475d9c6290b : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9e6561ce67434b449560e1554c5063a2 : SUCCESS in 43m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5198f7e197d34b92be31a968ed308a36 : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d57df4a1eb104235b12ecf8e6cbbf62c : SUCCESS in 1h 17m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e38cae7fb57742e69791f38ede403fdf : FAILURE in 1h 51m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/72d7fd51772a40b6b2ade483bb3c0c53 : SUCCESS in 1h 45m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b3027278f8d4137954c8e2e6ea89313 : SUCCESS in 1h 18m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ee13dfc7aaba4a4d9829f71e2d91c4ba : SUCCESS in 1h 14m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5dd696cee41b4b5885e0ebb372026c2a : SUCCESS in 1h 33m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2d02c80a59e47218b4ae03fc332c0ee : SUCCESS in 2h 31m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b304e17efe1b478bbd989fdb730184b9 : SUCCESS in 1h 30m 36s","accounts_in_message":[],"_revision_number":11},{"id":"325b2576830196d80da3cda93225f40c043c0f53","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-27 01:17:53.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-9/ns5_nfs : SUCCESS in 3h 4m 30s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-9/ns5_iscsi : SUCCESS in 2h 40m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"d59deb0ccd69b133ba335981c8b74aad23aabd44","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-27 01:49:29.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/56/766856/11/upstream-solidfire-check/cinder-solidfire-iSCSI/00afe6b/ : SUCCESS in 54m 26s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":11},{"id":"6afb5a6cd9f99afc778b50dd710288157df941b2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-27 01:51:30.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/56/766856/11/check/DellEMC_PowerStore_iSCSI/76d69fa/DellEMC_PowerStore_iSCSI/472 : SUCCESS in 57m 05s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/56/766856/11/check/DellEMC_PowerStore_FC/fb3c26e/DellEMC_PowerStore_FC/466 : SUCCESS in 48m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"74afdf9cd3ef2db13392f62f4cd7507670504930","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-27 02:10:11.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-10/ns5_nfs : SUCCESS in 3h 26m 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-10/ns5_iscsi : SUCCESS in 3h 5m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"61e7a4bbd475b56bd6833d7709d51c55f076d148","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-27 02:38:22.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-11/ns5_nfs : SUCCESS in 3h 24m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-11/ns5_iscsi : SUCCESS in 3h 5m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"44726aa2c02f5ef4db5adabfd7c021a08169812d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-27 03:58:50.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/56/766856/11/upstream-check/cinder-cDOT-iSCSI/42c99af/ : SUCCESS in 36m 36s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/56/766856/11/upstream-check/cinder-cDOT-iSCSI-IPv6/669d77d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/56/766856/11/upstream-check/cinder-cDOT-NFS/58a2dbe/ : FAILURE in 3h 01m 23s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/56/766856/11/upstream-check/cinder-cDOT-NFS-IPv6/4732a5e/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/56/766856/11/upstream-check/cinder-cDOT-FCP/c4eb287/ : SUCCESS in 56m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"894f9b8d8c8d19cd93d9af42037a6a6b5d43e621","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-27 06:01:10.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/11/check/DellEMC_Unity_FC/b61801c/DellEMC_Unity_FC/788 : SUCCESS in 1h 30m 27s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/11/check/DellEMC_Unity_iSCSI/7209565/DellEMC_Unity_iSCSI/750 : SUCCESS in 1h 37m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"c5d289471204556fdf19ef0a6a6bb1f53e69c431","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-27 06:37:32.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/11/check/DellEMC_XtremIO_FC/8beffcd/DellEMC_XtremIO_FC/876 : SUCCESS in 2h 02m 05s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/11/check/DellEMC_XtremIO_iSCSI/23065af/DellEMC_XtremIO_iSCSI/882 : SUCCESS in 1h 39m 17s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":11},{"id":"28bc89a6b813f43cf5682ee9525aed0667afe888","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-27 06:52:33.000000000","message":"Patch Set 11:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/766856-11/Lenovo-iSCSI : SUCCESS in 48:42.40\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/766856-11/Lenovo-FC : SUCCESS in 54:34.92","accounts_in_message":[],"_revision_number":11},{"id":"bb6c495f5e0484ed6d7079abc561769a3b4d0e11","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-27 06:53:57.000000000","message":"Patch Set 11:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/766856-11/PVME-iSCSI : SUCCESS in 49:52.91\n* PVME-FC http://os-logs.tristero.net/dell-ci/766856-11/PVME-FC : SUCCESS in 49:26.93","accounts_in_message":[],"_revision_number":11},{"id":"bc33f4a69950cb462ceec9f2dbf9da2fdfb4b094","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-27 07:01:26.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/56/766856/11/check/DellEMC_PowerFlex/a111c34/DellEMC_PowerFlex/447 : FAILURE in 2h 34m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"2f99c41d80ce55865ad11dc9860fef47baf3b6eb","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-27 07:40:07.000000000","message":"Patch Set 11:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/766856-11/HPMSA-iSCSI : FAILURE in 1:01:38\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/766856-11/HPMSA-FC : SUCCESS in 36:18.50","accounts_in_message":[],"_revision_number":11},{"id":"7fc12283395e91745ae301b0b3ac1282c6cc247c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-27 10:48:47.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/56/766856/11/check/DellEMC_SC_FC/9e43d1f/DellEMC_SC_FC/395 : FAILURE in 1h 54m 39s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/56/766856/11/check/DellEMC_SC_iSCSI/f9f021d/DellEMC_SC_iSCSI/392 : FAILURE in 2h 52m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":11},{"id":"85228f86369eef0eeaf7459b8a5272fbfcdfc78b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-27 11:53:25.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/56/766856/11/check/DellEMC_VNX_iSCSI/646cd08/DellEMC_VNX_iSCSI/455 : SUCCESS in 2h 57m 42s\n- DellEMC_VNX_FC http://publiclogs.emc.com/56/766856/11/check/DellEMC_VNX_FC/7d1b8eb/DellEMC_VNX_FC/474 : SUCCESS in 1h 21m 25s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":11},{"id":"6a526520f98293db64d05f64113745b84eb283e6","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-27 23:17:12.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/11/20210228055510/NEC-ISCSI : SUCCESS in 2h 18m 44s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/11/20210228055510/NEC-FC : SUCCESS in 2h 22m 00s","accounts_in_message":[],"_revision_number":11},{"id":"4fc500442276035235c59858e686e6cf0c4c16d3","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-02-28 14:57:32.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8b5d6f543a926bccf74df324c5c2aa078ddcfe74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-28 15:24:03.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c94b4cc845e68a79b1e0c05f7cb6ff4a69a0d9ef","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-28 15:26:49.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"15536073c9e4c9855cb2eb157bd12a0d7d2fd253","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-28 15:33:40.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/56/766856/12/check/dsvm-tempest-TOYOU-ACS5000-driver/68349d3 : FAILURE in 2m 04s","accounts_in_message":[],"_revision_number":12},{"id":"bc2249d7762e5e47300070eda72b502b35746af1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-28 15:37:04.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/56/766856/12/check/iscsi-driver-nimble/b27390f/b27390f.html : FAILURE in 11m 39s","accounts_in_message":[],"_revision_number":12},{"id":"e0c802b09e3ea8683eff292ac7beb154b14717f1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-28 16:06:45.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/12/check-not-vote/ext-cinder-zuul/dc751db : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":12},{"id":"fad9b935ebeed52495b0d817bf59b118c4f64837","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-28 16:45:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/56/766856/12/ : SUCCESS in 1h 14m 39s","accounts_in_message":[],"_revision_number":12},{"id":"240f78bc60c9655363c1a0cb343b088216aed9be","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-28 16:46:31.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/12/check/iscsi-driver-primera/5a9fe59/5a9fe59.html : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":12},{"id":"27e677716053edb10178d48824ea9c7bee572d0a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-28 16:58:17.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/12/check/dsvm-tempest-cisco-zonemanager-job/31fb6ee/ : SUCCESS in 1h 22m 17s","accounts_in_message":[],"_revision_number":12},{"id":"9ee5add254dea113b38a286e3e2240aafa81d0cc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-28 17:07:01.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/12/check/DellEMC_Unity_FC/214ca04/DellEMC_Unity_FC/790 : SUCCESS in 1h 19m 39s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/12/check/DellEMC_Unity_iSCSI/a1c9637/DellEMC_Unity_iSCSI/752 : SUCCESS in 1h 42m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"937c6ae02d24963f564c0002b95e1ecd2e686baa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-28 17:08:00.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/12/check/tempest-dsvm-full-focal-py3/4fa1dd6/ : SUCCESS in 1h 40m 18s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/12/check/devstack-plugin-ceph-dsvm-focal-py3/dcf5f45/ : SUCCESS in 1h 28m 07s","accounts_in_message":[],"_revision_number":12},{"id":"692e570b4cacf3dbfb7604edcae41ad2d94d9c35","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-28 17:13:22.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/766856/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6ac7f35/ : POST_FAILURE in 1h 48m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/766856/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2dd45f2/ : SUCCESS in 1h 29m 12s","accounts_in_message":[],"_revision_number":12},{"id":"604d0b7fe814bd6c5a852d4dac66fbf4e246a603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 17:22:00.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e20149edb3549928ba1afa79403cdad : FAILURE in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7edb7b071ac49498aca7a98deb54449 : FAILURE in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb69300c9f4040e287835a5a89283507 : FAILURE in 9m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a936d5a66b64deaaaed17d28e5445d7 : FAILURE in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0eb89ad86a344cfa74ca4a722198eda : FAILURE in 16m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972db70716b844ed8489faa6e47a5980 : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f61d61b557204b0f9ac29d0f40cf3818 : SUCCESS in 54m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01a0cf8d00954234b20b1c1167fcc110 : SUCCESS in 1h 01m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e407562db72484b987aa142eb8e3c75 : SUCCESS in 5m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/10bddde85596449ea70050acb3060e35 : FAILURE in 11m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/02897a95f06c460bb4661acbad0770c4 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a539121016a4b298ff6912a17542031 : SUCCESS in 7m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c88684b9afcc4bb3a0f3b599e264bd56 : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6de14ae227034da386e76ac28ecbf187 : SUCCESS in 37m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa128729ea3a4ae18c10676c4d18e96a : FAILURE in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a54afcbeec0a435c85732cdd29f97e2d : SUCCESS in 1h 31m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0cfa05ad8ae94cb1a2f8cb60fe48dd47 : FAILURE in 1h 27m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9bf06e0148d0455aa83a25d55f899fb2 : SUCCESS in 1h 12m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/13ad98b88704446eb5b39e616dc63403 : SUCCESS in 1h 05m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f3957858fae4c758840257ce5a36908 : FAILURE in 1h 15m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6721978da7be49a7853be6b4ea877827 : SUCCESS in 1h 28m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85a22710ec4743008abd61c3d415b0ca : SUCCESS in 1h 53m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e419edb0d62d4df5a3a94cb5a9455170 : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":12},{"id":"16a1cd99f8eae9cf5d444f59604a1d8c98479ca9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-28 17:28:36.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/12/check/DellEMC_XtremIO_FC/53023e5/DellEMC_XtremIO_FC/878 : SUCCESS in 2h 04m 04s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/12/check/DellEMC_XtremIO_iSCSI/1afed8e/DellEMC_XtremIO_iSCSI/884 : SUCCESS in 1h 39m 06s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":12},{"id":"b0376bdada76a638a165f19586a2c8a5aa113282","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-28 17:49:45.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/12/20210301002406/NEC-ISCSI : SUCCESS in 2h 25m 12s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/12/20210301002406/NEC-FC : SUCCESS in 2h 25m 37s","accounts_in_message":[],"_revision_number":12},{"id":"e18c874612a9072956637ba8e9c1a2fd6d8d07bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-28 18:03:21.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0e001ceb2231868d1fd321a2d858d977f3a3ab3e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-28 18:03:47.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"23aaffd6347fb782cb8a25bc37bad7d6e9b77760","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-28 18:04:12.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"aa570d3d6da0608d7a9c64fb390657bea65c74e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-28 18:18:09.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"31f197065fe6dbd0b0b05dd011dd9d12039c3ea2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-28 18:31:07.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/56/766856/14/check/iscsi-driver-nimble/2a60c24/2a60c24.html : FAILURE in 11m 38s","accounts_in_message":[],"_revision_number":14},{"id":"64131b8e444d614ef179c7725a908910d144d369","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-28 18:32:25.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-12/ns5_nfs : SUCCESS in 3h 7m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-12/ns5_iscsi : SUCCESS in 2h 58m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"792d98ad33cbcf0a091a031b30acc9c3475655c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-28 18:58:34.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/14/check-not-vote/ext-cinder-zuul/e81663c : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":14},{"id":"6891d1093b1b468cbb274ddeb092ada30d3c014c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-28 19:37:04.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/56/766856/14/ : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":14},{"id":"19ff3b0c6995c10d406ff3c0cc847d920e29cc1f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-28 19:40:12.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/14/check/iscsi-driver-primera/a539f5f/a539f5f.html : SUCCESS in 1h 20m 37s","accounts_in_message":[],"_revision_number":14},{"id":"1d1a1ebe5e536fc364cef2d0921f7f4e41594503","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-28 19:46:57.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/14/check/dsvm-tempest-cisco-zonemanager-job/21c2d2c/ : SUCCESS in 1h 18m 58s","accounts_in_message":[],"_revision_number":14},{"id":"a6da2f23d28dc1b7fb8fdfb7a5e101412e377adf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-28 19:58:37.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/14/check/DellEMC_Unity_FC/c319979/DellEMC_Unity_FC/792 : SUCCESS in 1h 27m 20s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/14/check/DellEMC_Unity_iSCSI/6f13afd/DellEMC_Unity_iSCSI/754 : SUCCESS in 1h 39m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"acb4eea1da9bd33a3f75b438753b3ccddd424c97","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-28 19:59:30.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/766856/14/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b80d889/ : SUCCESS in 1h 40m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/766856/14/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3f17f8b/ : SUCCESS in 1h 33m 21s","accounts_in_message":[],"_revision_number":14},{"id":"eb68e9dcd8413c4ca84a4eb29b149c1bed789083","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-28 20:03:27.000000000","message":"Patch Set 14:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/14/check/tempest-dsvm-full-focal-py3/06bb808/ : SUCCESS in 1h 39m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/14/check/devstack-plugin-ceph-dsvm-focal-py3/e0b62b7/ : SUCCESS in 1h 40m 41s","accounts_in_message":[],"_revision_number":14},{"id":"2a5d804d040b1568e4c7a75075eda547f74b47c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 20:08:29.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e58c619b56fb4929af5892bd4eda3ba2 : FAILURE in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1549a2dc7a848da97c4f8302eadd87b : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64bdc248ba3241a2ba2053056f985ffb : FAILURE in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60cc23d12003421c8940d004aa128b17 : FAILURE in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dae983dbfd840f3a71be7c98cdb5d43 : FAILURE in 12m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a36453e46142a5969a251f3ec6acb8 : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/37a6ab8115eb44ac825599250a31b282 : SUCCESS in 58m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8507b18c2b074b90ae16b5d256a61484 : SUCCESS in 1h 03m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b0e498e4067407c81e49dee025343f0 : SUCCESS in 6m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1fd4fbdefc45407ea7a0898a6996a8ee : FAILURE in 11m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9291ef80dd4d4bdc8ee8e983926b4d23 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e9f61818a974ba4826f041ed331a49e : SUCCESS in 5m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ec9730c710a478682f539d97d53fd1f : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/067b70d9b30f4e029ee5cc3ee22c1074 : SUCCESS in 42m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4dfda4ff134644cf82a1e0145f785581 : FAILURE in 7m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2ad7d11d72b4ac497f6cc5487412cd2 : SUCCESS in 1h 06m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/36eb0affdcbb491ba8a10bbaba66476b : FAILURE in 1h 24m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/448642ecbb6e43919ffce93d9e860428 : SUCCESS in 1h 22m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e674b3485f074340bf4ba771fbc6ce7b : SUCCESS in 1h 08m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a9cefad2a29b4c99baea8b1028647119 : SUCCESS in 1h 15m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/89a32990beb249d7990e9ebf6ec51b1f : SUCCESS in 1h 14m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86ab22f54b7941d6a7c7fd885cb57fce : SUCCESS in 1h 48m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e7c125206eb48d7b5693df6c4d3f2dd : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":14},{"id":"07e04042f3fa86b812d3a0a50c4e8012d795192a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-28 20:20:24.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/14/check/DellEMC_XtremIO_FC/853996d/DellEMC_XtremIO_FC/880 : SUCCESS in 2h 01m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/14/check/DellEMC_XtremIO_iSCSI/2aadf92/DellEMC_XtremIO_iSCSI/886 : SUCCESS in 1h 36m 06s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":14},{"id":"6f70c0041cb4836fc83be4368623764a82be9d80","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-28 20:22:50.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/56/766856/14/check/dsvm-tempest-TOYOU-ACS5000-driver/33068e0 : FAILURE in 1h 00m 23s","accounts_in_message":[],"_revision_number":14},{"id":"79424927bc65cf9756f9204712c5b6971f349bfc","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-28 20:26:25.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/13/20210301030326/NEC-ISCSI : SUCCESS in 2h 11m 44s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/13/20210301030326/NEC-FC : SUCCESS in 2h 22m 57s","accounts_in_message":[],"_revision_number":13},{"id":"8896a36d048a1d85b7b12f4bf27244010c9075af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-02-28 20:29:56.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"c2c6b3507775b6d778b4a093fbd188646e369ce8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-28 20:30:24.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"bb46e34034f20995218573a9b3c342edfbfa93f0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-28 20:30:48.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"930cd67b3293fef24f83ab43c6e361d37d86de11","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-28 21:12:15.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/15/check-not-vote/ext-cinder-zuul/371cb31 : SUCCESS in 41m 16s","accounts_in_message":[],"_revision_number":15},{"id":"ab62d799c8104759b22ad2c9bc1864254283ea09","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-28 21:26:17.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-13/ns5_nfs : SUCCESS in 3h 22m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-13/ns5_iscsi : SUCCESS in 3h 19m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"095671ffd3b606287c5727a1b3722b986bebf165","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-28 21:37:30.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/56/766856/15/check/dsvm-tempest-TOYOU-ACS5000-driver/6af52a9 : FAILURE in 1h 06m 16s","accounts_in_message":[],"_revision_number":15},{"id":"d3a6556455ee4ce589fad0e321b87836fef6baa3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-28 21:47:10.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-14/ns5_nfs : SUCCESS in 3h 13m 14s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-14/ns5_iscsi : SUCCESS in 3h 6m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"13c0c57efe8dd16f50c576af159fd6f61ad02e78","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-28 21:47:17.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/56/766856/15/check/iscsi-driver-primera/c4a8b99/c4a8b99.html : SUCCESS in 1h 15m 56s","accounts_in_message":[],"_revision_number":15},{"id":"357722999537acb8cd6363640813294917ea07b8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-28 21:48:43.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/56/766856/15/ : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":15},{"id":"9b593a4c1b12435e4603919f204aa7e804f52616","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-28 21:55:23.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/15/check/dsvm-tempest-cisco-zonemanager-job/d27e6b3/ : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":15},{"id":"73db46409a76599fd1bb4c5cdc7b2abd267f9e70","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-28 22:13:07.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/766856/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b9a2eba/ : SUCCESS in 1h 42m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/766856/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/02f638f/ : SUCCESS in 1h 37m 54s","accounts_in_message":[],"_revision_number":15},{"id":"a7a4e1d59b92c03c3936880776103c7e593b92a8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-28 22:15:13.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/15/check/tempest-dsvm-full-focal-py3/99397b8/ : SUCCESS in 1h 31m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/15/check/devstack-plugin-ceph-dsvm-focal-py3/3f7a102/ : SUCCESS in 1h 40m 49s","accounts_in_message":[],"_revision_number":15},{"id":"a09f85287d84abc2070e4a30d9b9cebea1051e5a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-28 22:19:03.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/15/check/DellEMC_Unity_FC/6376806/DellEMC_Unity_FC/793 : SUCCESS in 1h 22m 07s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/15/check/DellEMC_Unity_iSCSI/5f7fe22/DellEMC_Unity_iSCSI/755 : SUCCESS in 1h 48m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"4f5cfcf0da69368defebed234cf92d17db3f2d98","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-28 22:32:43.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/15/check/DellEMC_XtremIO_FC/5393ea1/DellEMC_XtremIO_FC/881 : SUCCESS in 2h 02m 15s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/15/check/DellEMC_XtremIO_iSCSI/dd860d1/DellEMC_XtremIO_iSCSI/887 : SUCCESS in 1h 38m 30s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"9222c92410dc6c89b29b28923f39f97befb3e084","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-28 22:47:23.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/14/20210301052630/NEC-ISCSI : SUCCESS in 2h 09m 55s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/14/20210301052630/NEC-FC : SUCCESS in 2h 20m 51s","accounts_in_message":[],"_revision_number":14},{"id":"749e870b24b7f84c3e330e691fd951132cd72958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 22:51:06.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2134c5d3b55e456aaab04d3ae54548e5 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fda8d0e70684cb7a6a8a28fccb6013c : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc491af10a674ba1a7be4ef422abbe86 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7fb5ed4aa7c4284846728f3dc1a80d2 : SUCCESS in 8m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af20a72ad7864719b6dbd12f227bd84c : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f50616c277b4d69b465bbcbdcbe6af6 : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/2755930421584157be922c710529b4a8 : SUCCESS in 1h 04m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7f1239493fb49c398db964e50e2f2ae : SUCCESS in 50m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff19ef0d1a94c77b3749ee4edbdac28 : SUCCESS in 5m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c62207825cbf4e468aae99a27f4673a1 : SUCCESS in 11m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ac8f81ca44547ee8f612baee9112579 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/240a9240d68a4ef2b919976d1cf6a1ac : SUCCESS in 6m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb735465e5fe4ccbbc6ac9d17b23bfd4 : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a0481a87e81a4fae98a576e81dcb3015 : SUCCESS in 44m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77063b620e5848fd8de9a8931a864463 : FAILURE in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/44dfbb4854be40f0a10ac789626aacf3 : SUCCESS in 1h 22m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e26ba5cc2918449fb2410aa8dba83009 : FAILURE in 1h 27m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0572e067df2348beb3758f22313e2973 : SUCCESS in 1h 09m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6fa96ebb3c6343f6ae23b3f8ecce6a91 : SUCCESS in 1h 15m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/19ba70c2312443d3acc84077a1b33997 : SUCCESS in 56m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d244c5b2d1814e56854104fb7966cd14 : SUCCESS in 1h 04m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35b9e128ce9b48628636a497edbcdb27 : SUCCESS in 2h 14m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e478d78f612b474fb77a1a6a2afae487 : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":15},{"id":"9e2a7262b58f8e17943edf15b3c01583e1427ace","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-28 23:35:47.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-15/ns5_nfs : SUCCESS in 3h 5m 11s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-15/ns5_iscsi : SUCCESS in 3h 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"05105949f333f5fdfecf2fcf836441831285afad","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-28 23:49:34.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/56/766856/15/check/DellEMC_PowerStore_iSCSI/a268359/DellEMC_PowerStore_iSCSI/473 : SUCCESS in 57m 54s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/56/766856/15/check/DellEMC_PowerStore_FC/3bf3282/DellEMC_PowerStore_FC/467 : SUCCESS in 53m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":15},{"id":"39bf47e5d20c8d7d298e37f7f1de882ae2389b95","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-28 23:51:32.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/56/766856/15/upstream-solidfire-check/cinder-solidfire-iSCSI/0c1886b/ : FAILURE in 58m 44s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"09faea07e9d714d8b7647cea2c4d8dc7f9af4e21","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-01 00:01:50.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/56/766856/15/upstream-check/cinder-cDOT-iSCSI/60c8817/ : SUCCESS in 40m 14s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/56/766856/15/upstream-check/cinder-cDOT-iSCSI-IPv6/8217c55/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/56/766856/15/upstream-check/cinder-cDOT-NFS/c759745/ : SUCCESS in 37m 03s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/56/766856/15/upstream-check/cinder-cDOT-NFS-IPv6/320e222/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/56/766856/15/upstream-check/cinder-cDOT-FCP/2180149/ : SUCCESS in 1h 09m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"2f4c5b6ce174ab971ebdec63d7e84453bd51ab96","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-01 00:43:55.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/56/766856/15/check/DellEMC_Unity_FC/289f389/DellEMC_Unity_FC/794 : SUCCESS in 1h 24m 57s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/56/766856/15/check/DellEMC_Unity_iSCSI/cc69523/DellEMC_Unity_iSCSI/756 : SUCCESS in 1h 51m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"a88d314ff416ef47645b8b89f2610c8bf385f3f3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-01 00:53:04.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/15/check/DellEMC_XtremIO_FC/5e44dae/DellEMC_XtremIO_FC/882 : SUCCESS in 2h 01m 21s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/15/check/DellEMC_XtremIO_iSCSI/b7a1794/DellEMC_XtremIO_iSCSI/888 : SUCCESS in 1h 40m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":15},{"id":"b319289bcc6a2790d58581114fd67af6a09a820f","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-01 01:08:01.000000000","message":"Patch Set 15:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/766856-15/PVME-iSCSI : SUCCESS in 51:18.78\n* PVME-FC http://os-logs.tristero.net/dell-ci/766856-15/PVME-FC : SUCCESS in 48:52.46","accounts_in_message":[],"_revision_number":15},{"id":"ee88c446c1f02c9cade74765dd4c77ca161c03c4","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-03-01 01:08:05.000000000","message":"Patch Set 15:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/766856-15/HPMSA-iSCSI : FAILURE in 1:03:15\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/766856-15/HPMSA-FC : SUCCESS in 37:01.59","accounts_in_message":[],"_revision_number":15},{"id":"093a7a63df0fe1747a5f6ef9a770a8893367d835","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-01 01:08:24.000000000","message":"Patch Set 15:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/766856-15/Seagate-iSCSI : SUCCESS in 49:52.18\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/766856-15/Seagate-FC : SUCCESS in 47:12.23","accounts_in_message":[],"_revision_number":15},{"id":"b000f917a3a57bc8972f287b99f07681f440aa4e","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-01 01:08:26.000000000","message":"Patch Set 15:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/766856-15/Lenovo-iSCSI : SUCCESS in 51:09.89\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/766856-15/Lenovo-FC : SUCCESS in 48:03.87","accounts_in_message":[],"_revision_number":15},{"id":"ce70110d451f06717acb739af777dcb1beaae69c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-01 01:33:31.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/56/766856/15/check/DellEMC_PowerFlex/fe53e5f/DellEMC_PowerFlex/448 : SUCCESS in 2h 41m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"e3ecd0aaba8a02b69e61380e068fcae56a46b59e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-01 01:43:18.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/56/766856/15/check/DellEMC_SC_FC/684b282/DellEMC_SC_FC/396 : SUCCESS in 1h 56m 56s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/56/766856/15/check/DellEMC_SC_iSCSI/b86c05c/DellEMC_SC_iSCSI/393 : FAILURE in 2h 51m 42s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":15},{"id":"b3537f5efc87512b6c9fb1e6747722c825a3cacb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-01 02:02:14.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/56/766856/15/check/DellEMC_VNX_iSCSI/e17b8fb/DellEMC_VNX_iSCSI/456 : SUCCESS in 3h 10m 38s\n- DellEMC_VNX_FC http://publiclogs.emc.com/56/766856/15/check/DellEMC_VNX_FC/49689c2/DellEMC_VNX_FC/475 : SUCCESS in 1h 19m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":15},{"id":"fa87569101762da28417a76be1e8105fa2a18927","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-01 03:32:11.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/15/20210301074728/NEC-ISCSI : SUCCESS in 2h 14m 08s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/15/20210301074728/NEC-FC : SUCCESS in 4h 44m 41s","accounts_in_message":[],"_revision_number":15},{"id":"52defb5b89456023954ed8351d7942df03cac2bb","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-01 15:11:43.000000000","message":"Patch Set 15: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":15},{"id":"4b85777bf1113cb15b502a13ca19d4bdf9a319a1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-03 16:30:56.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/56/766856/15/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c3b1314/c3b1314.html : FAILURE in 18m 18s","accounts_in_message":[],"_revision_number":15},{"id":"69f1c6a87af3ce411999432e6fc4f04b3f1f0da4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-12 22:30:16.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)\n\nCommit message needs an update to reflect the current patch.","accounts_in_message":[],"_revision_number":15},{"id":"98974c8cfc450d59bfaa18c359a505bac31fae20","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-12 22:32:29.000000000","message":"Patch Set 15: Review-Priority+1","accounts_in_message":[],"_revision_number":15},{"id":"834a8821fa7c8490302d0c4d38f392559e842485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-13 11:50:50.000000000","message":"Uploaded patch set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"7c0e5aa45684f543ebd75e5f2357a876da53d850","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-13 11:52:57.000000000","message":"Patch Set 16:\n\n(1 comment)\n\nSorry I missed to correct the commit message - should be good now.","accounts_in_message":[],"_revision_number":16},{"id":"8cc1035dc52df1dc9c61b24b9cb1df1eb81f12c0","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-13 11:57:23.000000000","message":"Patch Set 16:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/56/766856/16/check/dsvm-tempest-TOYOU-ACS5000-driver/d52c729 : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":16},{"id":"14b4a5615f238072fda04641444a4cca5a07f2a2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-13 13:18:45.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/16/check/dsvm-tempest-cisco-zonemanager-job/2b9ef36/ : FAILURE in 1h 21m 58s","accounts_in_message":[],"_revision_number":16},{"id":"93e118b91e69e997e02734018d9942442ce73689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 13:46:38.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bd252c685674ee181604491dc193cd8 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9a02367a504e5bbb9885c91e060113 : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d9c4a74f4c1483aa024df2b7f5dde1e : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44a28e0a388c4fd3ad3a287278b52948 : SUCCESS in 13m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64171eb008514f2186f4c4fd42371e7b : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03eb662f836412eacf33d749ec2553a : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/af95382e1b2c4a24b816c7fd63565631 : SUCCESS in 58m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a9adead974144767aad7390d9cffcdc0 : SUCCESS in 1h 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53db79f1d63b47d29eb4877aec761a14 : SUCCESS in 9m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d0a2700d51642f4b10546f5ca4d3f84 : SUCCESS in 12m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1c13826b32749e68bee75a5d4adc121 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df2c412e6afd4b458572a4e21c567ea1 : SUCCESS in 5m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d282a04cfb54cc284cc24287e828457 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/83ee0f6cc4054d7b9faf79c4e9a39fc3 : SUCCESS in 43m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2e18e66d7fd4e07afffee6f77c22699 : FAILURE in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70a7a8b463e940afbe76913e06af8cde : SUCCESS in 1h 18m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f22c4dc70695449db5403187319f169a : FAILURE in 1h 41m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a0373806550f46139120f8811a4a0089 : SUCCESS in 1h 12m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c07e3bc1f8743c484a38df2a7f17c50 : SUCCESS in 55m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1784b9aaec9247aca34d8c41274f2ba6 : SUCCESS in 1h 18m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/459035ff90044e95b611ac650c580a5a : FAILURE in 1h 31m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0105df8c0987452d86fc837fefdb5a0b : SUCCESS in 1h 53m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/718fb1ce1b684ed8869719d1f0c59d80 : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":16},{"id":"22842c446cf6f7257113b5e2bc479274214c064a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-13 14:44:49.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/56/766856/16/check/DellEMC_PowerFlex/4ee156e/DellEMC_PowerFlex/567 : FAILURE in 57m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"1d1673d2ebbe406d669aa5d93b98b6ab5619b0c6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-13 14:52:58.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/56/766856/16/check/DellEMC_PowerStore_iSCSI/cd9a295/DellEMC_PowerStore_iSCSI/599 : SUCCESS in 1h 05m 47s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/56/766856/16/check/DellEMC_PowerStore_FC/fc7b2cb/DellEMC_PowerStore_FC/593 : SUCCESS in 54m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":16},{"id":"5b4d55d965d6498fb2f993132e4c1ff413e5e26f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-13 14:57:16.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/766856/16/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/77edfe4/ : SUCCESS in 3h 00m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/766856/16/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/37ce177/ : TIMED_OUT in 3h 05m 06s","accounts_in_message":[],"_revision_number":16},{"id":"440d23623c9aec010c02debb47b1929d7fb29a37","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-13 15:32:36.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/74e7628a51cf4d009aa18444f0a53daa : POST_FAILURE in 1h 31m 02s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b672e93e46a449a7b8e397d822f19a3a : TIMED_OUT in 2h 32m 45s","accounts_in_message":[],"_revision_number":16},{"id":"5dedd1ffb4ba20271387a046f6bf42900efc4850","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-13 23:50:03.000000000","message":"Patch Set 16:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/766856-16/Seagate-iSCSI : SUCCESS in 46:09.93\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/766856-16/Seagate-FC : SUCCESS in 42:13.84","accounts_in_message":[],"_revision_number":16},{"id":"5a9c5f34ab988594fe1736331e81ce9aee50f1a6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-14 02:12:09.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-16/ns5_nfs : SUCCESS in 6h 27m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-16/ns5_iscsi : SUCCESS in 6h 17m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":16},{"id":"4de8140dd540b212f5e9f042bd30bdfc57be6ac2","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-14 04:27:59.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/56/766856/16/check/dsvm-tempest-zadara-driver/d8753ac : SUCCESS in 1h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"75427ddd189278e3e12807927298e8e7de413cc2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-14 20:05:54.000000000","message":"Patch Set 16:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/16/check-not-vote/ext-cinder-zuul/7716ac2 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":16},{"id":"065b659bbe8cde6f506a5334663007153011fb98","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-14 22:24:12.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/16/check/DellEMC_XtremIO_FC/2fa49bc/DellEMC_XtremIO_FC/1004 : SUCCESS in 2h 01m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/16/check/DellEMC_XtremIO_iSCSI/3c740c1/DellEMC_XtremIO_iSCSI/1006 : SUCCESS in 1h 44m 02s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":16},{"id":"8f5c92fc65e7c8f36d3df2df542f9f998ff3b8c9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-15 02:46:25.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/56/766856/16/check/DellEMC_VNX_iSCSI/1897bbc/DellEMC_VNX_iSCSI/559 : SUCCESS in 3h 11m 03s\n- DellEMC_VNX_FC http://publiclogs.emc.com/56/766856/16/check/DellEMC_VNX_FC/961debb/DellEMC_VNX_FC/587 : SUCCESS in 1h 27m 29s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":16},{"id":"4700ad996a8755d131159c077393c9cebdec7b92","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-15 13:19:53.000000000","message":"Patch Set 16: Code-Review-1\n\n(2 comments)\n\nSomehow I missed the mention of the config option in the release note.","accounts_in_message":[],"_revision_number":16},{"id":"30984d76230296b8736339b89bac3cd8f98921ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-15 13:44:26.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"f48e5745ac5e389d226093913b6ad9c48b908775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-15 13:46:06.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"5198d9855decc36edb503f8690bb4bbb9bdea444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-15 13:47:22.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"575909157feb6d9ef66fd4d2e788a60df1aeea96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-15 13:48:41.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"37f9bbd8676e4e42fd98ab09cf6d623139641d53","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-15 13:50:34.000000000","message":"Patch Set 20:\n\n(2 comments)\n\nI fixed the release notes - hopefully.","accounts_in_message":[],"_revision_number":20},{"id":"01e5cfbc54f95c98c1a3dd4c3a73397b06e3c6a5","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-15 14:30:28.000000000","message":"Patch Set 20:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/56/766856/20/check/dsvm-tempest-TOYOU-ACS5000-driver/64fc423 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":20},{"id":"2e90df04b7cd78ac370f4a5f113ae0666cfff995","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-15 14:31:21.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/20/check-not-vote/ext-cinder-zuul/ac0f296 : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":20},{"id":"042a688c3491d86c1aa1a01feddb8fc4563d3b90","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-15 15:14:41.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/20/check/dsvm-tempest-cisco-zonemanager-job/68c2db1/ : FAILURE in 1h 21m 39s","accounts_in_message":[],"_revision_number":20},{"id":"7d5131bd355fa6f252cfe0cdec24d642c02d12b1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-15 15:55:31.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/766856/20/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/449bd34/ : FAILURE in 2h 05m 16s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/766856/20/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c752027/ : FAILURE in 1h 56m 35s","accounts_in_message":[],"_revision_number":20},{"id":"e64e727726ea24e1d0bde5ed0aa868b9ccc9d932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 16:07:13.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b0cafc4aab24e8dbf68e6a4c1f16f09 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfcc208083f549d3979b8ef8f666e26c : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3ec30365dd44fc0b8df267cfe240a1e : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f022b4c2b39349a3a2287e912760c594 : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14df66a89c694d5c92271ad69c99a33b : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9838c4f248e846dfb1f9e1cae15a0fa9 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/d7c1031c72c44f60a9ecac1f60a52675 : SUCCESS in 1h 08m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b983aca75748478f993e164bc0dd5e26 : SUCCESS in 1h 09m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a9b268a98dd4bb3b38897a1e8aa7228 : SUCCESS in 5m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aff3da7150684e26b20faf6db4388c44 : SUCCESS in 12m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba25f677930640028d98b47f15dfaf76 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/606f947d481a49bbbb609b3af35219a1 : SUCCESS in 6m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4223993b5cae464798c23aa38e2ac4d6 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8bd16ade69eb4b0fadede71852ae3029 : SUCCESS in 40m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/40ebc7dad1eb4cd5a09cd046be9b7170 : FAILURE in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61a9390985804851b749fc7a8e5ef133 : SUCCESS in 1h 22m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a8db2c75699242ee9a4802d831072e58 : FAILURE in 1h 32m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f3f4ac18b73d4444a1a760a4049b19d6 : SUCCESS in 1h 30m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88bfd9f968aa4e6f9872d60f6d01d104 : SUCCESS in 1h 26m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a8f72e06af7142bc8cbbbf60f0abce11 : SUCCESS in 1h 21m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a207e6c4e2704d4eb64ad9a96623282a : FAILURE in 1h 37m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4b616adebff4af6a62c7c035181d90a : FAILURE in 2h 16m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4bdf84bcd3b4d4abbac013a6ff53050 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":20},{"id":"73ceb4e7eb70cd532205b3ca994e510089b9c8f8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-15 16:13:15.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/56/766856/20/check/dsvm-tempest-zadara-driver/e6f841c : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"6210d91e823ce43aac989adc3226a68a862b9dda","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-15 16:19:24.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"08025a269e98782f52bb0e416f31471a4c899bb8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-15 17:19:32.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/766856/766856/20/check-not-vote/ext-cinder-zuul/47938d1 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":20},{"id":"b653f7401acbf6c812163d9fa5c772fdf9582855","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-15 17:43:43.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/20/check/DellEMC_XtremIO_FC/cd6859e/DellEMC_XtremIO_FC/1013 : FAILURE in 2h 02m 10s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/20/check/DellEMC_XtremIO_iSCSI/c42446b/DellEMC_XtremIO_iSCSI/1015 : SUCCESS in 1h 42m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":20},{"id":"862fc8c8c2fd5557cb8a981df01d833ec6c07868","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-15 18:11:54.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/56/766856/20/check/dsvm-tempest-cisco-zonemanager-job/5f36456/ : FAILURE in 1h 23m 57s","accounts_in_message":[],"_revision_number":20},{"id":"ecf7b830051834e2f0b745e065b1dda669b5420b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-15 18:23:57.000000000","message":"Patch Set 20:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/20/check/tempest-dsvm-full-focal-py3/d82bcf7/ : TIMED_OUT in 3h 10m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/56/766856/20/check/devstack-plugin-ceph-dsvm-focal-py3/b68a324/ : FAILURE in 2h 11m 38s","accounts_in_message":[],"_revision_number":20},{"id":"a0c6857fde73eaae24b0e83418101c6f6b065f2a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-15 18:26:13.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/766856/20/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8946c1d/ : SUCCESS in 1h 59m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/766856/20/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/120f2eb/ : SUCCESS in 2h 05m 20s","accounts_in_message":[],"_revision_number":20},{"id":"b3a51a7c59b2f08cdcb0425f91f4267779a05a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:45:34.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1000298cbd5943c7a24572cc4cd1a541 : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a19b9e96c01426fa7955f4e994cadc5 : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34c9b7bfa3ef4d07a2255ccb75aa1904 : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2ab580da3ea42c6aba64d9b069b9036 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a007007948e43aaa1ae7bfd11779610 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb3638ed0434bc49956c5b2e1a1a634 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/832e8a8c0526450d8860a5a58c29a1aa : SUCCESS in 1h 02m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ec92f89442843e0baa93a55ca744699 : SUCCESS in 1h 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f6b62b625aa49d48f5fcf44102ee4eb : SUCCESS in 7m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3233b26255a2485b92db42d2997084b3 : SUCCESS in 18m 35s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/434ef07db32640769aff6e9f0f80e089 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5fd4d76fedc459e8f8698f8a2a5e1f4 : SUCCESS in 7m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41eeb41b35b94bf4bf582b13b3058c8c : SUCCESS in 9m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4c678e202d0f46b8a8fdc582128ffe4f : SUCCESS in 43m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/237f4b17897340d1a3a52b187dbc8e57 : FAILURE in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc3ebc5179a0499b8bdf728c8705f12f : SUCCESS in 1h 32m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b41d815df76a43ddbe4f68994c5dd66e : FAILURE in 1h 26m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/11331341850a4f568da7b2f17a2d66a3 : SUCCESS in 1h 47m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/881d67eefd0f40f5873cb5a48355730a : SUCCESS in 1h 22m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f138e7b69c0b45378934a1d5d01c7a61 : SUCCESS in 1h 26m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cc8e35825ee24015b7ac7c8a004b1edf : FAILURE in 1h 14m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af2b50b1a0a34db3ace41ae48451fb1c : SUCCESS in 2h 07m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7eeeaf1ed8042f9a36ab337e12b7f11 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":20},{"id":"8282244d7e8e2ed422341ede65b83a16dcacea6c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-03-15 19:14:14.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-56-766856-20/1615833937 : SUCCESS in 28m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":20},{"id":"63c705ea692d410fe3f32788514ccf2df9f554bf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-15 19:23:35.000000000","message":"Patch Set 17:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-17/ns5_nfs : FAILURE in 42m 50s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-17/ns5_iscsi : FAILURE in 5h 13m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":17},{"id":"a15bb498f1d5c029fece3461d3b457794190f9ac","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-15 19:45:48.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/56/766856/20/upstream-solidfire-check/cinder-solidfire-iSCSI/469d878/ : FAILURE in 58m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":20},{"id":"8bc30bb461919bf74efb54db4ba7b3dc14a635ea","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-15 20:01:17.000000000","message":"Patch Set 18:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-18/ns5_nfs : FAILURE in 5h 31m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-18/ns5_iscsi : SUCCESS in 5h 9m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":18},{"id":"c6250d2fe1df62a651d67cc54ac886397a9b9a48","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-15 20:03:06.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/56/766856/20/check/DellEMC_PowerStore_iSCSI/cd1afbe/DellEMC_PowerStore_iSCSI/608 : SUCCESS in 56m 48s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/56/766856/20/check/DellEMC_PowerStore_FC/c19dfa3/DellEMC_PowerStore_FC/602 : SUCCESS in 51m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":20},{"id":"90ac92582e96508627c4d1b6761bf71582ea1d46","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-15 20:07:56.000000000","message":"Patch Set 19:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-19/ns5_nfs : FAILURE in 5h 18m 10s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-19/ns5_iscsi : SUCCESS in 5h 16m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":19},{"id":"c1f5a23cc1a8a7f67d86851ebb9a9682d8541cb6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-15 20:10:09.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/56/766856/20/check/DellEMC_PowerFlex/8e8b272/DellEMC_PowerFlex/576 : FAILURE in 57m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"e08b91af1ec4dabbdf970d9291c9494ad1968d1b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-15 20:26:15.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/56/766856/20/upstream-check/cinder-cDOT-iSCSI/841be60/ : SUCCESS in 39m 22s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/56/766856/20/upstream-check/cinder-cDOT-iSCSI-IPv6/997f578/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/56/766856/20/upstream-check/cinder-cDOT-NFS/882ebb8/ : FAILURE in 1h 39m 35s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/56/766856/20/upstream-check/cinder-cDOT-NFS-IPv6/907b18d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/56/766856/20/upstream-check/cinder-cDOT-FCP/607f2cc/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"bc1959fc60b7951a469ef51bfe7bad0d676c815e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-15 20:32:02.000000000","message":"Patch Set 20:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-20/ns5_nfs : FAILURE in 46m 16s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-20/ns5_iscsi : FAILURE in 5h 22m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":20},{"id":"0325f4666e24182fdb0f2f1adb0bb87ea6583078","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-15 21:06:01.000000000","message":"Patch Set 20:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/766856-20/PVME-iSCSI : SUCCESS in 58:16.33\n* PVME-FC http://os-logs.tristero.net/dell-ci/766856-20/PVME-FC : SUCCESS in 51:01.74","accounts_in_message":[],"_revision_number":20},{"id":"87d89b3b94c3cfb6dd2759e40bb5be77ab01bbd2","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-15 22:16:31.000000000","message":"Patch Set 20:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/766856-20/Lenovo-iSCSI : SUCCESS in 51:21.30\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/766856-20/Lenovo-FC : SUCCESS in 45:08.70","accounts_in_message":[],"_revision_number":20},{"id":"99c617a8c6a33646faf3b13a957d46eacf3c9694","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-15 22:26:08.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/56/766856/20/check/dsvm-tempest-zadara-driver/50f1956 : SUCCESS in 1h 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"c75450dcce82a1934c2e525946777fe1e347f23c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-16 01:50:38.000000000","message":"Patch Set 20:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-20/ns5_nfs : SUCCESS in 4h 46m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-20/ns5_iscsi : SUCCESS in 4h 46m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":20},{"id":"4e83b567f2e112d6e511f1c3715260acced1ff25","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-16 05:24:29.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/56/766856/20/check/DellEMC_VNX_iSCSI/80fa125/DellEMC_VNX_iSCSI/567 : SUCCESS in 3h 21m 39s\n- DellEMC_VNX_FC http://publiclogs.emc.com/56/766856/20/check/DellEMC_VNX_FC/1b28507/DellEMC_VNX_FC/595 : SUCCESS in 1h 25m 30s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":20},{"id":"a9e99431329115623bd3fe0b5a1ef144581d85dc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-16 05:58:46.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/56/766856/20/check/DellEMC_XtremIO_FC/34335ff/DellEMC_XtremIO_FC/1019 : SUCCESS in 2h 02m 08s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/56/766856/20/check/DellEMC_XtremIO_iSCSI/e8d29d6/DellEMC_XtremIO_iSCSI/1021 : SUCCESS in 1h 42m 21s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":20},{"id":"19d8d9f4ef431d5d0bb5cd883292a2c59c2983ad","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-16 06:28:26.000000000","message":"Patch Set 20:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-20/ns5_nfs : SUCCESS in 5h 9m 4s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-56-766856-20/ns5_iscsi : SUCCESS in 4h 58m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":20},{"id":"17fc13a21ac60387af6011af95ec9aa78eddabf7","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-16 14:14:17.000000000","message":"Patch Set 20: Code-Review+1\n\nLooks good to me, but it looks like we\u0027re going to need https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/778960 merged in order to test it.","accounts_in_message":[],"_revision_number":20},{"id":"eb901f285cfc22a480f694f95ee75d0799961668","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-03-16 15:18:37.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"2673c7a7324450ea4f0fcdd82558471744f419e0","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-17 04:55:12.000000000","message":"Patch Set 20: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":20},{"id":"482b356a0fc7111833a5bc2b5c526c344ab3404f","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-17 08:59:42.000000000","message":"Patch Set 20:\n\nThanks for your kind reviews and +1s :-)\n\n\nIf this is merged at some point - would you accept a backport / cherry-pick of this change for Victoria and Ussuri?","accounts_in_message":[],"_revision_number":20},{"id":"b46095c89b681eee79e66eadb955bcd83a4c3a4a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-17 13:31:14.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/20/20210317200156/NEC-ISCSI : SUCCESS in 2h 23m 30s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/56/766856/20/20210317200156/NEC-FC : SUCCESS in 2h 29m 14s","accounts_in_message":[],"_revision_number":20},{"id":"1004127ec0e27266fb39f4870333c469d72089b6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-17 20:18:05.000000000","message":"Patch Set 20: Code-Review+2 Review-Priority+1\n\nLGTM.  Raising review priority because it\u0027s been taking so long to move this along.  Thanks to Christian for being willing to revise it to remove the config option.","accounts_in_message":[],"_revision_number":20},{"id":"25019e26fad3c6faf8daf64736c2aa8f61e088a1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-17 20:23:35.000000000","message":"Patch Set 20:\n\n@Christian: go ahead and propose a backport to victoria after this merges.  It\u0027s a small change isolated to a single driver, which is good, but it has the words \"add support for feature\" in the commit message subject, which is kind of a red flag for the OpenStack stable maint team.  But we\u0027ll find out what the cinder stable maint team thinks once you\u0027ve proposed the backport.","accounts_in_message":[],"_revision_number":20},{"id":"e04761cdd365cab31bc8e8022afafddc14914d38","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-18 09:02:48.000000000","message":"Patch Set 20:\n\n\u003e Patch Set 20:\n\u003e \n\u003e @Christian: go ahead and propose a backport to victoria after this merges.  It\u0027s a small change isolated to a single driver, which is good, but it has the words \"add support for feature\" in the commit message subject, which is kind of a red flag for the OpenStack stable maint team.  But we\u0027ll find out what the cinder stable maint team thinks once you\u0027ve proposed the backport.\n\n@Brian: \n\nThen what is left for this change to be merged to master? It has plently of +1 or +2 for code review? :-)\n\nI\u0027d love to jump onto proposing a backport to Victoria and also Ussuri as soon as that\u0027s done.","accounts_in_message":[],"_revision_number":20},{"id":"9bc865719137a0e2e4468e5b79dd62f5abc532f6","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-24 17:53:15.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1\n\n(1 comment)\n\nOne typo in the release note but I don\u0027t think that is worth holding this up.  Can be fixed separately.","accounts_in_message":[],"_revision_number":20},{"id":"2beb36411729782c76a9226c49a4469b2f980ff1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 17:53:52.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"1dc15dcaf61150f4ead3455715aa9f4e5a75f09f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 20:39:24.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b61f5beed85541e4987a4c939c385c77 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e0df7fea01412faee5e46cb962d971 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f70ac76e0a14d5b87f292cb57fe3f21 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8a1889465414a95bf5ed5156052f212 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15732ff971784798b380e0b3e060271c : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/6a84d4d089114828b471820befc3fc65 : SUCCESS in 1h 08m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e092fdb55b8043d0aa5ec0dab9ddb0a7 : SUCCESS in 1h 38m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c474260e55c4549968b946f66973dd6 : SUCCESS in 7m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/070bd9057a714079b4db33e05317900e : SUCCESS in 1h 38m 11s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/548ab2f94d4a4746af66fd27702d99dc : SUCCESS in 1h 46m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7e64e8ff1214da68f485bcfe718c804 : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":20},{"id":"92e952e12f3fbffa246e0da0ba18ec90fea87ed9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 20:39:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"7ceda1af60ef146c7e54c634d54c6af623ce2bfc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 20:42:47.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dffe987fa8d4c9f90be15748279c67b : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d0f98d8c8454c02a4505456125ce141 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"0e8807388209086ee60e1c9f11a48ed6191e4178","revisions":{"47b2c3cff6b52834a8a5fa4f594ce39b91639a93":{"kind":"REWORK","_number":1,"created":"2020-12-13 14:32:17.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/1"}}},"commit":{"parents":[{"commit":"f845f9021936c7bcfb61ebf88dbe3cfaf39d7f9f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f845f9021936c7bcfb61ebf88dbe3cfaf39d7f9f"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:31:08.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for images stored in Ceph","message":"Add support for RBD fast-diff feature for images stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the gerneration of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47b2c3cff6b52834a8a5fa4f594ce39b91639a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47b2c3cff6b52834a8a5fa4f594ce39b91639a93"}]},"branch":"refs/heads/master"},"937ff063de88d05e764a095d2f274701b512091b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-13 14:44:37.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/2"}}},"commit":{"parents":[{"commit":"f845f9021936c7bcfb61ebf88dbe3cfaf39d7f9f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f845f9021936c7bcfb61ebf88dbe3cfaf39d7f9f"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:44:37.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for images stored in Ceph","message":"Add support for RBD fast-diff feature for images stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/937ff063de88d05e764a095d2f274701b512091b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/937ff063de88d05e764a095d2f274701b512091b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6193b89550ab87d3d005cbc5fcfaced8ce0b14a5":{"kind":"REWORK","_number":3,"created":"2021-01-14 13:32:14.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/3"}}},"commit":{"parents":[{"commit":"f19a92064ac8acd5e7d141d83fe90cbef4402f46","subject":"LVM: Support only LVM 2.02.107+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f19a92064ac8acd5e7d141d83fe90cbef4402f46"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-01-14 13:31:54.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for images stored in Ceph","message":"Add support for RBD fast-diff feature for images stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the gerneration of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6193b89550ab87d3d005cbc5fcfaced8ce0b14a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6193b89550ab87d3d005cbc5fcfaced8ce0b14a5"}]},"branch":"refs/heads/master"},"34c15d0d6cfbeda0409d150e0fadefd627af3557":{"kind":"REWORK","_number":4,"created":"2021-01-17 16:49:41.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/4"}}},"commit":{"parents":[{"commit":"f19a92064ac8acd5e7d141d83fe90cbef4402f46","subject":"LVM: Support only LVM 2.02.107+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f19a92064ac8acd5e7d141d83fe90cbef4402f46"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-01-17 16:48:04.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for images stored in Ceph","message":"Add support for RBD fast-diff feature for images stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the gerneration of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34c15d0d6cfbeda0409d150e0fadefd627af3557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34c15d0d6cfbeda0409d150e0fadefd627af3557"}]},"branch":"refs/heads/master"},"6249ed16a1efbe612b7248789a40e3da015b2ed0":{"kind":"REWORK","_number":5,"created":"2021-01-17 16:55:43.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/5"}}},"commit":{"parents":[{"commit":"f3d1168856667df2bec30d655ed6dcd4ccf77fe1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3d1168856667df2bec30d655ed6dcd4ccf77fe1"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-01-17 16:55:33.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for images stored in Ceph","message":"Add support for RBD fast-diff feature for images stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the gerneration of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6249ed16a1efbe612b7248789a40e3da015b2ed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6249ed16a1efbe612b7248789a40e3da015b2ed0"}]},"branch":"refs/heads/master"},"2fa8b9b5a734ffa9f0e1f4be55896eaa33b69e18":{"kind":"REWORK","_number":6,"created":"2021-01-18 12:20:16.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/6"}}},"commit":{"parents":[{"commit":"f3d1168856667df2bec30d655ed6dcd4ccf77fe1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3d1168856667df2bec30d655ed6dcd4ccf77fe1"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-01-18 12:20:08.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for images stored in Ceph","message":"Add support for RBD fast-diff feature for images stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the gerneration of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2fa8b9b5a734ffa9f0e1f4be55896eaa33b69e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2fa8b9b5a734ffa9f0e1f4be55896eaa33b69e18"}]},"branch":"refs/heads/master"},"f6bef28ef2827c28faee79043f6c10b1592a74cd":{"kind":"REWORK","_number":7,"created":"2021-01-22 13:58:37.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/7"}}},"commit":{"parents":[{"commit":"f3d1168856667df2bec30d655ed6dcd4ccf77fe1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3d1168856667df2bec30d655ed6dcd4ccf77fe1"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-01-22 13:58:24.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for images stored in Ceph","message":"Add support for RBD fast-diff feature for images stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the gerneration of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6bef28ef2827c28faee79043f6c10b1592a74cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6bef28ef2827c28faee79043f6c10b1592a74cd"}]},"branch":"refs/heads/master"},"9aa577726a9f4064cb8a319e6dd0176b464f0faa":{"kind":"REWORK","_number":8,"created":"2021-02-25 08:18:30.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/8"}}},"commit":{"parents":[{"commit":"f3d1168856667df2bec30d655ed6dcd4ccf77fe1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3d1168856667df2bec30d655ed6dcd4ccf77fe1"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-02-25 08:07:54.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9aa577726a9f4064cb8a319e6dd0176b464f0faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9aa577726a9f4064cb8a319e6dd0176b464f0faa"}]},"branch":"refs/heads/master"},"1c319ac7f76e9bacf5d315887128ec9ed38f9efd":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2021-02-26 22:12:35.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/9"}}},"commit":{"parents":[{"commit":"f3d1168856667df2bec30d655ed6dcd4ccf77fe1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3d1168856667df2bec30d655ed6dcd4ccf77fe1"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-02-26 22:11:48.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature, if supported by RBD, for \n    backup images stored via Ceph RBD. It is enabled together with \n    EXCLUSIVE_LOCK and OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c319ac7f76e9bacf5d315887128ec9ed38f9efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c319ac7f76e9bacf5d315887128ec9ed38f9efd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ff516a13f6e428af990792c6a3f6474580d54823":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-02-26 22:13:18.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/10"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-02-26 22:13:18.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature, if supported by RBD, for \n    backup images stored via Ceph RBD. It is enabled together with \n    EXCLUSIVE_LOCK and OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff516a13f6e428af990792c6a3f6474580d54823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff516a13f6e428af990792c6a3f6474580d54823"}]},"branch":"refs/heads/master","description":"Rebase"},"939d29e4444a7bc1e42f5800cd631026bd61da94":{"kind":"REWORK","_number":11,"created":"2021-02-26 22:16:18.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/11"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-02-26 22:16:09.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature, if supported by RBD, for \n    backup images stored via Ceph RBD. It is enabled together with \n    EXCLUSIVE_LOCK and OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/939d29e4444a7bc1e42f5800cd631026bd61da94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/939d29e4444a7bc1e42f5800cd631026bd61da94"}]},"branch":"refs/heads/master"},"657f4d0546634bc432155e8be259d03596d9cbfc":{"kind":"REWORK","_number":12,"created":"2021-02-28 15:24:03.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/12"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-02-28 15:23:49.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/657f4d0546634bc432155e8be259d03596d9cbfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/657f4d0546634bc432155e8be259d03596d9cbfc"}]},"branch":"refs/heads/master"},"ad5ca63fad3d5a59751240aed3b9aae603b35243":{"kind":"REWORK","_number":13,"created":"2021-02-28 18:03:21.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/13"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-02-28 18:02:59.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad5ca63fad3d5a59751240aed3b9aae603b35243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad5ca63fad3d5a59751240aed3b9aae603b35243"}]},"branch":"refs/heads/master"},"024b34bda2cce59c5805be22607bc69705da1934":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-02-28 18:18:09.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/14"}}},"commit":{"parents":[{"commit":"85975fb2b63866ba6e216e621b84d571b9dbe90e","subject":"Merge \"mypy: annotate volume manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85975fb2b63866ba6e216e621b84d571b9dbe90e"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-02-28 18:18:09.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/024b34bda2cce59c5805be22607bc69705da1934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/024b34bda2cce59c5805be22607bc69705da1934"}]},"branch":"refs/heads/master","description":"Rebase"},"bac384674a6fba733aba8dcc5dec9db309a8f0bd":{"kind":"REWORK","_number":15,"created":"2021-02-28 20:29:56.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/15"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-02-28 20:29:49.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change adds a config switch \u0027backup_ceph_fast_diff\u0027 to\n    enable the FAST_DIFF feature for backup images stored via Ceph RBD.\n    This feature is enabled together with EXCLUSIVE_LOCK and OBJECT_MAP\n    as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bac384674a6fba733aba8dcc5dec9db309a8f0bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bac384674a6fba733aba8dcc5dec9db309a8f0bd"}]},"branch":"refs/heads/master"},"d339cf3005dc1389f535a52077901b7995001158":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2021-03-13 11:50:50.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/16"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-13 11:49:50.000000000","tz":60},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature for backup images stored\n    via Ceph RBD. It is enabled together with EXCLUSIVE_LOCK and\n    OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d339cf3005dc1389f535a52077901b7995001158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d339cf3005dc1389f535a52077901b7995001158"}]},"branch":"refs/heads/master"},"48b9c2df733b640b74e3c8a27c2f93fbaa1d79e5":{"kind":"REWORK","_number":17,"created":"2021-03-15 13:44:26.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/17"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-15 13:44:17.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature for backup images stored\n    via Ceph RBD. It is enabled together with EXCLUSIVE_LOCK and\n    OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48b9c2df733b640b74e3c8a27c2f93fbaa1d79e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48b9c2df733b640b74e3c8a27c2f93fbaa1d79e5"}]},"branch":"refs/heads/master"},"990d01dbdea656ed38ace5acab57cf434d5d1151":{"kind":"REWORK","_number":18,"created":"2021-03-15 13:46:06.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/18"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-15 13:45:35.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature for backup images stored\n    via Ceph RBD. It is enabled together with EXCLUSIVE_LOCK and\n    OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/990d01dbdea656ed38ace5acab57cf434d5d1151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/990d01dbdea656ed38ace5acab57cf434d5d1151"}]},"branch":"refs/heads/master"},"57f38a2a38f69e17d7317ec88e48804ec1c36248":{"kind":"REWORK","_number":19,"created":"2021-03-15 13:47:22.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/19"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-15 13:47:13.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature for backup images stored\n    via Ceph RBD. It is enabled together with EXCLUSIVE_LOCK and\n    OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57f38a2a38f69e17d7317ec88e48804ec1c36248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57f38a2a38f69e17d7317ec88e48804ec1c36248"}]},"branch":"refs/heads/master"},"0e8807388209086ee60e1c9f11a48ed6191e4178":{"kind":"REWORK","_number":20,"created":"2021-03-15 13:48:41.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/56/766856/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/766856/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/766856/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/766856/20"}}},"commit":{"parents":[{"commit":"c6b2ea4253b332454dfac62feee187ba19ef84ba","subject":"Merge \"Hitachi: Wait until the volume can be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b2ea4253b332454dfac62feee187ba19ef84ba"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2020-12-13 14:07:10.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2021-03-15 13:48:28.000000000","tz":0},"subject":"Add support for RBD fast-diff feature for backups stored in Ceph","message":"Add support for RBD fast-diff feature for backups stored in Ceph\n\n    This change enables the FAST_DIFF feature for backup images stored\n    via Ceph RBD. It is enabled together with EXCLUSIVE_LOCK and\n    OBJECT_MAP as required.\n\n    With FAST_DIFF enabled, the generation of diffs between images and\n    snapshots as well as determining the actual data usage of a snapshot\n    is speed up significantly.\n\nCloses-Bug #1907964\n\nChange-Id: I5e35de28007558267f9215cbac03a38d7c74612c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e8807388209086ee60e1c9f11a48ed6191e4178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e8807388209086ee60e1c9f11a48ed6191e4178"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
