)]}'
{"id":"openstack%2Fcinder~767012","triplet_id":"openstack%2Fcinder~stable%2Fussuri~I10581ccf96e063f000f9a763adca66f3da72f60e","project":"openstack/cinder","branch":"stable/ussuri","topic":"check-l-c","hashtags":[],"change_id":"I10581ccf96e063f000f9a763adca66f3da72f60e","subject":"DNM: check status of ussuri zuul check","status":"ABANDONED","created":"2020-12-14 20:04:00.000000000","updated":"2020-12-18 01:50:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"669c66f307f647bc7960d21c773e22a7f366f234","_number":767012,"virtual_id_number":767012,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-14 22:47:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 20:25:49.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 20:30:01.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-14 21:09:17.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-14 21:19:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 21:37:10.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 22:10:35.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 22:32:38.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 22:33:40.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 22:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-14 22:48:21.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"1196c5e60048052f9fab42edb9c058f7f2f69e64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-14 20:04:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4b5757511c4c2f7711fb8a49b1f630aa5f1c1f8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-14 20:25:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/12/767012/1/ : RETRY_LIMIT in 2m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8d6009b11f6fc0200dbb3d2099e5dc0d5c16979a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-14 20:30:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/12/767012/1/check/dsvm-tempest-cisco-zonemanager-job/5c2029c/ : FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"763b2f719f8ef7df394c7d9b6dd5a9bff6d1280c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-14 21:09:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767012/1/iscsi/ : SUCCESS in 1h 04m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/767012/1/smb/ : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c8b9f2ded9f3497153ce1b5eabe4e595b6fe4c96","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-14 21:19:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/12/767012/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4bca6fd/4bca6fd.html : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":1},{"id":"04a9dd3bf59543cdd7e9c959ec1ed9bd0f7410a4","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-14 21:37:10.000000000","message":"Patch Set 1:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/12/767012/1/check/dsvm-tempest-zadara-driver/02a8996 : FAILURE in 1h 32m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9f9186d88271e0a7997303a52b62be341294aa7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-14 22:10:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/12/767012/1/20201215050404/NEC-ISCSI : SUCCESS in 2h 05m 13s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/12/767012/1/20201215050404/NEC-FC : SUCCESS in 2h 06m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ce1c086d73e02f8500f1bfe2efa3559d61ea5476","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-14 22:32:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/12/767012/1/check/tempest-dsvm-full-bionic/3fb63bd/ : SUCCESS in 1h 55m 19s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/12/767012/1/check/tempest-dsvm-full-bionic-py3/0a04a83/ : SUCCESS in 1h 52m 25s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/12/767012/1/check/devstack-plugin-ceph-dsvm-bionic-py3/80258fa/ : SUCCESS in 2h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2e68f9c9afdba2ed69baa2d14a152b488ed0769e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-14 22:33:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/12/767012/1/Cinder-lioadm : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/767012/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4b48a42ab98adcac00b4c99e6cec55a2ed6f78e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 22:47:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4092542d1dc46059d1d08663ab687f8 : FAILURE in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b190ad34321d4aeeac0ef8196330a8fc : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab9280f4d1ee496791b17fbac335cd51 : SUCCESS in 9m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e35281043bcb4ce78e088a0f84adab10 : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789ff3d0e97f4d939b4c96c35c161936 : SUCCESS in 13m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/a395b0e9c167477cbea1767d4e368097 : SUCCESS in 1h 26m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5710c86a0d764f97b42e91693a43def4 : SUCCESS in 1h 30m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d5ee1fc9f5c4de19128dc03a01933d2 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf710daf93514c408ad552cb0ea22067 : SUCCESS in 5m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a67e48f220674d82ad59d4363a0de75c : SUCCESS in 7m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/833593d6ff684f159d967f8658938280 : SUCCESS in 43m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/549c19972fc6490785fb2a9663c34378 : SUCCESS in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/18c49fb590f04b0fb1590e95435edf96 : SUCCESS in 1h 14m 35s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a592c4791ec94d26b40dc36b196903ab : SUCCESS in 1h 42m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb107383b7cd48009666f8c5056ade25 : SUCCESS in 1h 48m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e13e919489b4c71be3e1a7dd70b850b : SUCCESS in 1h 28m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c28ce95cbbc14775a46e7d0e83a818e6 : FAILURE in 1h 21m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f45b29d3f17f43d98590f5cdfccd7515 : SUCCESS in 2h 35m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ac703344e4b4f99a8826633f1536ed7 : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":1},{"id":"aaf3f0682a3400171595bfdd4e7456cb44e6a024","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-14 22:48:21.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-767012-1/ns5_nfs : SUCCESS in 2h 43m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-767012-1/ns5_iscsi : SUCCESS in 2h 39m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"60b2f5d719ddd93dfd942f1421cbcebce45f0ca9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-15 04:18:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/767012/1/Cinder-tgtadm : FAILURE\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/12/767012/1/Cinder-lioadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"669c66f307f647bc7960d21c773e22a7f366f234","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-18 01:50:29.000000000","message":"Abandoned\n\nno longer needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd5b2f04e6b7f3468e1f495323bff0aa51866006","revisions":{"dd5b2f04e6b7f3468e1f495323bff0aa51866006":{"kind":"REWORK","_number":1,"created":"2020-12-14 20:04:00.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/12/767012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/767012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/767012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/767012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/767012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/767012/1"}}},"commit":{"parents":[{"commit":"2f0d236440b6816ae2e50ae10192777fcf4117dd","subject":"Merge \"Fix volume rekey during clone\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f0d236440b6816ae2e50ae10192777fcf4117dd"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-12-14 20:03:16.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-12-14 20:03:16.000000000","tz":-300},"subject":"DNM: check status of ussuri zuul check","message":"DNM: check status of ussuri zuul check\n\nChange-Id: I10581ccf96e063f000f9a763adca66f3da72f60e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd5b2f04e6b7f3468e1f495323bff0aa51866006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd5b2f04e6b7f3468e1f495323bff0aa51866006"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
