)]}'
{"id":"openstack%2Fcinder~767106","triplet_id":"openstack%2Fcinder~stable%2Frocky~Ie597e778e3efa61bfd98eaaa92730b5050195e7a","project":"openstack/cinder","branch":"stable/rocky","topic":"py27-requirements","hashtags":[],"change_id":"Ie597e778e3efa61bfd98eaaa92730b5050195e7a","subject":"[stable-only] Cap bandit to v1.6.2 and fix constraints","status":"MERGED","created":"2020-12-15 09:39:25.000000000","updated":"2021-06-08 20:46:34.000000000","submitted":"2021-06-04 17:16:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767106-py27-requirements","meta_rev_id":"2e85efce9a59360cf45b8e1fcff9a7ea0325fe93","_number":767106,"virtual_id_number":767106,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 17:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-03 12:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-06-04 15:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-06-04 15:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-06-03 12:15:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"CC":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 10:53:14.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-15 11:17:16.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-15 12:40:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 14:34:52.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 15:35:46.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-07 14:28:01.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2021-01-18 19:57:37.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 15:34:02.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2021-04-13 17:21:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-04-13 18:42:28.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-04-13 19:25:34.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-13 21:25:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-06-02 21:44:39.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-06-02 21:53:24.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-03 12:15:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-04 15:33:36.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-06-08 20:46:34.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"3077df255dd06095f8dc019df1b6cde05bb40ce9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 09:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"069c1f2949d44c71eb20b8a6a25f5fa64015abbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 09:47:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9cbeca96990ce276eae431c69364e32113119a36","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 09:51:00.000000000","message":"Topic set to py27-requirements","accounts_in_message":[],"_revision_number":2},{"id":"19bec44d745ca3c6558d312792ea32ca828241ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 10:02:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"17853081da243c6409e832260122da192a4a82d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 10:34:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"21769f8778d0d010c6a131a6d5da3ebc0762e0fd","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-15 10:53:14.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767106/4/iscsi/ : FAILURE in 10m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/767106/4/smb/ : FAILURE in 8m 58s","accounts_in_message":[],"_revision_number":4},{"id":"56e9c5b389cb21d7a9307dbc548427ed5cd6a36b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-15 11:17:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/4/check/dsvm-tempest-cisco-zonemanager-job/74c3572/ : FAILURE in 16m 24s","accounts_in_message":[],"_revision_number":4},{"id":"46553c97c914bef840b3c7e911690d339a3135b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 12:40:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afbf4ffa9754e2bb6fd433dbb670fb3 : FAILURE in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cdd0bbad2584e728462f454c822c9dd : SUCCESS in 12m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c1380efd30a844ae905b264b038d53c2 : FAILURE in 4m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a6191124a63f4f73a070e75096f913ac : SUCCESS in 11m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c848061faaeb42c4b86adaac16bb6ebe : SUCCESS in 3m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5ff2585dccd7445197f429b8a1bdac9e : SUCCESS in 1h 15m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e82ba7d6634490f8e8c219ddb774eb8 : SUCCESS in 1h 25m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e368d629d40a4983be040eef63c11bd3 : SUCCESS in 5m 59s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f7abb1a40ed34a659619f0f8fe751934 : SUCCESS in 1h 55m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/deb77a17c0cd4aa7afbf95f954bd2919 : FAILURE in 3m 44s (non-voting)\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/37b433a5ff35468ab2bef0ce041a4d6b : FAILURE in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8580d33747b4790be01e657378700b1 : FAILURE in 3m 56s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2a41a666c1b470d8c357cd2059f8908 : FAILURE in 9m 57s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/02bfd14593c045279dc859f7921aa5b2 : FAILURE in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a987fcc02ba2484d9cd4463d3f48a936 : FAILURE in 13m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/95bba45b476a44f7b2d5a79436c09e5a : FAILURE in 28m 17s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/62bc68196b9b44bb9335830cfad48f47 : FAILURE in 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c66f16016fc38b82152f60b711698e349e2c3690","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-15 14:34:52.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/3/check/tempest-dsvm-full-bionic-py3/44b13c6/ : SUCCESS in 1h 53m 32s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/3/check/devstack-plugin-ceph-dsvm-bionic-py3/d577532/ : FAILURE in 59m 39s","accounts_in_message":[],"_revision_number":3},{"id":"e67028ffa318cdf9fb7ef6cd806d4e781d7583e8","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-15 15:35:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/06/767106/4/20201215214039/NEC-FC : SUCCESS in 2h 25m 50s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/06/767106/4/20201215214039/NEC-ISCSI : SUCCESS in 2h 55m 05s","accounts_in_message":[],"_revision_number":4},{"id":"2e96a21ff297a9388b8be6af86dc8cbdd85fc88e","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-01-07 14:28:01.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks Ivan, LGTM","accounts_in_message":[],"_revision_number":4},{"id":"bb98b58a5ad856be4e7740d11813a7bb50011b30","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-01-18 18:34:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8af16c94b86bc51507d7a527f8888d288968fd88","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-18 19:03:23.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767106/4/iscsi/ : FAILURE in 13m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/767106/4/smb/ : FAILURE in 17m 09s","accounts_in_message":[],"_revision_number":4},{"id":"55f8dd8adbd80ba36151db7e71b9aef5a12be836","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-18 19:44:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/4/check/dsvm-tempest-cisco-zonemanager-job/fed76ac/ : FAILURE in 18m 39s","accounts_in_message":[],"_revision_number":4},{"id":"000bbd6cd8b0524e6f5e046252fb55dd279e521e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-18 19:57:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/06/767106/4/check/dsvm-tempest-zadara-driver/493c8b3 : SUCCESS in 1h 22m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f0368aa962ba2abdbf16dced0a5b21d38d5da0b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-18 22:48:11.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/4/check/tempest-dsvm-full-bionic-py3/d945550/ : SUCCESS in 1h 45m 07s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/4/check/devstack-plugin-ceph-dsvm-bionic-py3/f4d30e8/ : TIMED_OUT in 2h 37m 23s","accounts_in_message":[],"_revision_number":4},{"id":"2f71a418f8e3f2e84110dbb125a3c463912c423a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 01:50:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3504eec8fa46e097d4b7851b7beb8f : FAILURE in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4bd80c8c2bba43fe9187c2815f47daf0 : SUCCESS in 8m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d11f121784744acb91efd9b0bfe80c71 : FAILURE in 4m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/595d41aeb63a4d1392010203a2126f65 : SUCCESS in 9m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7e37e83fc53d479992df4e13c310312c : SUCCESS in 3m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6daed02c884848579d7cd07a6f4e3f63 : SUCCESS in 1h 32m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2da84b18279a41cab3c61b94bc3ce4cf : SUCCESS in 1h 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a16beb9e467b4746b9e9cd72bed62098 : SUCCESS in 7m 27s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/baae295f7e71477183fcf32ef1acbf32 : SUCCESS in 1h 42m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/858fc2cf78b447b8afc1904dcd7610fc : FAILURE in 3m 55s (non-voting)\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7236d4f745a041019154d60294b29ca7 : FAILURE in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b698c59752744875b8a6e55738caddd6 : FAILURE in 4m 00s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1932f208a064acebd2d42bf7a32d25f : FAILURE in 4m 54s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/2ad3e8f1542e4486bf703130e6d35fc0 : FAILURE in 4m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8ebea5db4eaf484fa5081fdddcc58e69 : FAILURE in 19m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7fd81b25a4f4c8f981fc33a20e5f38f : FAILURE in 28m 29s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d9350fe946142d0bd040acc70a1b0c6 : FAILURE in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7789bc59b8784b9b56cdac25973be16501cff6eb","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-19 15:34:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"01bcb055a149dbbf30ae391a912536fe438a6200","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-19 15:38:23.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767106/4/iscsi/ : FAILURE in 46s\n- cinder-smb http://cloudbase-ci.com/cinder/767106/4/smb/ : FAILURE in 47s","accounts_in_message":[],"_revision_number":4},{"id":"07f5e447eb7dffc61ebabd6e57856d1931ccabf9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-19 18:34:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/06/767106/4/check/dsvm-tempest-zadara-driver/f534526 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b92bf199b3b438de8b58acef9eba6c0e3aa7a0b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-19 18:43:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/4/check/dsvm-tempest-cisco-zonemanager-job/8af20b4/ : FAILURE in 20m 33s","accounts_in_message":[],"_revision_number":4},{"id":"5aa29e69fcffdba2113b7f483be5787185436645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 18:57:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06794a863f6c4764a62c6de0e71de777 : FAILURE in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a251dcdcd7a48ae818b026d031aee33 : SUCCESS in 10m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/97eb1ab84d25444cb0da34a67b16a292 : FAILURE in 4m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7f561d3e53b849f4a29492b414716761 : SUCCESS in 8m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac1b350f5e884b8ea38673f9f1eb017f : SUCCESS in 3m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/097f1d3e730a46d3bb97478fc8235de6 : SUCCESS in 1h 30m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/016c229eca8d46e9a07addc586174c14 : FAILURE in 1h 02m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/917b3dafe67b49dcb196fee32f49a20e : SUCCESS in 7m 00s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1d8a911402b94de1b2f6dae275235c98 : FAILURE in 1h 05m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/39fa77b530c842329c9eadf5a34b51a6 : FAILURE in 4m 49s (non-voting)\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ab8cd5ba6c2d45eb8384c4d9911c0a25 : FAILURE in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a294434ceb74189ab09e484414b4f73 : FAILURE in 4m 40s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a5b8e3bdb0994b7c8e9c33a3d7b1700a : FAILURE in 3m 55s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/ee3330db7bb14c17be474ded33179449 : FAILURE in 4m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2f61ee60caba44cc9e9ccdab0278f10a : FAILURE in 17m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee96600a7255490481fb9e5a170b6a17 : FAILURE in 27m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fd501ef23bee40e1b3b9023c52df0452 : FAILURE in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0a0c04602781efb9237e0372eb9fd30950cd75c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-19 21:26:04.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/4/check/tempest-dsvm-full-bionic-py3/077d624/ : SUCCESS in 1h 47m 33s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/4/check/devstack-plugin-ceph-dsvm-bionic-py3/cbf90df/ : FAILURE in 1h 36m 40s","accounts_in_message":[],"_revision_number":4},{"id":"f22687f9a90ce9c5481f98877302ad784b1e8d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-13 17:21:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d50599909ca3c88bb9aae7a704b99359e584e839","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-13 17:49:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/5/check/dsvm-tempest-cisco-zonemanager-job/ff3fae8/ : FAILURE in 22m 22s","accounts_in_message":[],"_revision_number":5},{"id":"4df7f4062aad35f84cdf7dec772e68da9b0d7e18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-13 18:18:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6f419eeb9817f090f965c88773ddfb5fdfb7c8da","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-13 18:42:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/06/767106/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/912b3b0/ : RETRY_LIMIT in 3m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/06/767106/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a01f57a/ : RETRY_LIMIT in 3m 30s","accounts_in_message":[],"_revision_number":6},{"id":"95b78a257d33136f01c9d842765406246f4b742d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-13 18:45:52.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/6/check/dsvm-tempest-cisco-zonemanager-job/41561fc/ : FAILURE in 21m 48s","accounts_in_message":[],"_revision_number":6},{"id":"241d79daf1932215db6aa1d3ba3ec582e4a2f583","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-13 19:25:34.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/06/767106/6/check/iscsi-driver-primera/7922295/7922295.html : FAILURE in 1h 02m 11s","accounts_in_message":[],"_revision_number":6},{"id":"6e2f6604596d0fd1553e4744502c1d1e5341c37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 20:00:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727a3b0b502b481eba3fbf3554a6306c : SUCCESS in 8m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a45ef46b166e4b13a2450579d8d7fc8b : SUCCESS in 10m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3fbe3b69d3f94d03b9b6debe599b86b4 : SUCCESS in 16m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f431cd9ad9fd44d8bef4d82eadcbf774 : SUCCESS in 9m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b99de8d8a1f441edbf437df2c6bdd559 : SUCCESS in 5m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/423d327a076b40589fca6e9ad581e912 : SUCCESS in 1h 22m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7a34e44e48684352ac574770b1353ded : SUCCESS in 59m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4431e4ad3bf842ab89b688a625d1a69f : SUCCESS in 1h 23m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df45e51a9b3e4877ad422f46ea63b5f1 : SUCCESS in 8m 48s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb2d1774cff94eeb8701f2f68676db0c : FAILURE in 1h 34m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc28acba98e0422c8c23ae3240c728a4 : SUCCESS in 9m 59s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/c72e023d11134ae8b8d5fcd34bdd7b6d : SUCCESS in 1h 06m 22s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5be7616c811d449c8620cb807db22cec : SUCCESS in 8m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63602b5ad2694d3aa4b3aad64bdaeb9d : SUCCESS in 12m 53s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9026085482d646b5bb0c7eb301a17998 : SUCCESS in 5m 21s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/ec0614ecb0fd4eac88c752361a452b96 : SUCCESS in 10m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f4f3e132c5b5414cb103212c03789f2a : FAILURE in 11m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f25a2d908666430f8a0599a3fe04d1a7 : SUCCESS in 27m 51s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/52a4f5ed22bf4637b736594070ef7310 : SUCCESS in 1h 08m 36s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/15ccad4ff6404569868ebc1fe617c298 : SUCCESS in 6m 00s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/54e65033095242b58b17d2e9aa142026 : SUCCESS in 1h 01m 14s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/40981c204b0c4330b6f6a21b2705e3ef : SUCCESS in 1h 12m 17s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/fc3eafdddcb04a68bd60c3ab45dbd097 : FAILURE in 12m 11s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/39d3d23d642d48b2820fb94dceb46dbc : SUCCESS in 1h 08m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ecd6ae582e66c6db7685876359486e39bb90e959","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-13 21:25:54.000000000","message":"Patch Set 6:\n\nI think I\u0027d prefer https://review.opendev.org/c/openstack/cinder/+/785696 which also removes the lower-constraints job.\n\nWe seem to have the same failures in cinder-tempest-dsvm-lvm-lio-barbican in the network scenario tests.  I wonder whether we need to be using a different tempest tag or turning off a network feature or something.","accounts_in_message":[],"_revision_number":6},{"id":"2a3e9280cd97b09800ab5219aaf2b9a27d1fa664","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-13 21:42:16.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/6/check/tempest-dsvm-full-bionic-py3/eba3e41/ : SUCCESS in 1h 49m 06s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/a1c3fd1b6dac43e2ab258ab62059b628 : POST_FAILURE in 1h 17m 47s","accounts_in_message":[],"_revision_number":6},{"id":"aae70196b947daeb5f231fb559afbdfb9f7a959f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-14 02:32:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/06/767106/6/check/iscsi-driver-nimble/56ffb55/56ffb55.html : FAILURE in 1h 06m 45s","accounts_in_message":[],"_revision_number":6},{"id":"03cbc9e6bc024ae30f47317c3d43716b2cb83483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-14 09:49:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c6b210e6b0622769d6242f2afca93952967e0f33","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-14 10:04:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/06/767106/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7a4ddd1/ : RETRY_LIMIT in 3m 37s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/06/767106/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8f7c463/ : RETRY_LIMIT in 3m 30s","accounts_in_message":[],"_revision_number":7},{"id":"3e63dc2bb767dcb908293cd85a8769d43f1c843d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-14 10:05:15.000000000","message":"Patch Set 7:\n\n\u003e I think I\u0027d prefer https://review.opendev.org/c/openstack/cinder/+/785696 which also removes the lower-constraints job.\n\nI understand and accept that. But... :) I\u0027d rather merge this patch to unblock the gate (this patch fixes the lower constraints too) and *then* remove the lower-constraints job with your patch if you think it is still needed. o:)\n\n(With this patch the lower constraints are a bit closer to a working lower constraints as at least the unit tests are working with the corrected constraints. Of course, I don\u0027t know whether it is really useful for someone. But anyway, it still feels a bit better to have a somewhat (better) working lower constraints for cinder. Even if we drop the job afterwards)\n\nAnyway, I accept if you don\u0027t want the extra round of patch. :)\n\n\u003e We seem to have the same failures in cinder-tempest-dsvm-lvm-lio-barbican in the network scenario tests.  I wonder whether we need to be using a different tempest tag or turning off a network feature or something.\n\nIf I\u0027m not mistaken the failure is the same that was fixed in stein with this patch I068cf1e9618d305b5a9383c283bfa0f120bfe905, but now we have it in cinder-tempest-dsvm-lvm-lio-barbican.\nLet\u0027s see if I have sorted it out with the new patch set.","accounts_in_message":[],"_revision_number":7},{"id":"fe09c92f34991104561b92d55132d8ed6b2b1bda","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-14 10:06:39.000000000","message":"Patch Set 7:\n\nAnd sorry @Ivan for kidnapping your patch, I hope it\u0027s not a problem o:)","accounts_in_message":[],"_revision_number":7},{"id":"d9ef6dd4a64a6cd9ac95b25151b0a83d549ba3b3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-14 10:16:54.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/7/check/dsvm-tempest-cisco-zonemanager-job/94ab4b7/ : FAILURE in 21m 23s","accounts_in_message":[],"_revision_number":7},{"id":"719dd9ffcfbabf708c6400b6142746909a6f7a16","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-14 10:47:27.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/06/767106/7/check/iscsi-driver-primera/1489360/1489360.html : FAILURE in 56m 28s","accounts_in_message":[],"_revision_number":7},{"id":"9615f1fc63e9d69723d548beb64f395e7471dc21","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-04-14 10:53:23.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/06/767106/7/check/iscsi-driver-nimble/805b7e0/805b7e0.html : FAILURE in 1h 02m 27s","accounts_in_message":[],"_revision_number":7},{"id":"d78b220fcd131d6688ffdf9f8799e753cf89464e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 11:11:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2be8fd54ea6411582d11b3c6bc1b885 : SUCCESS in 9m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91d91c8fd6c54019ab7fcf5235cd6aaa : SUCCESS in 8m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/172a393a3a4e4c2c9d726b9359697c33 : SUCCESS in 9m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/09cad2fa7a974a42b81d45d6b83d1d2e : SUCCESS in 9m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/670ea33cefb04432ba2d1db428af1fe8 : SUCCESS in 3m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/530da4c5f19c4d63b1841865ffc33abf : SUCCESS in 1h 14m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3433725cd16b479cb2837ce752ec422b : FAILURE in 14m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5e5c5d75d1449739ae7ab8bd2b1c999 : SUCCESS in 1h 18m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/012d5096734f47bcbc935c1f238a7e45 : SUCCESS in 6m 11s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae5c8976350f4206a8a5dca7b2896b35 : FAILURE in 10m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e604884acbd74ee383d22a34b601dcec : SUCCESS in 9m 01s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/09553f3455464068babee64a51296e6a : SUCCESS in 47m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/9f06dae99e9142139916d16e12574052 : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f949328bf7a4ccfaef79a8914dce334 : SUCCESS in 11m 31s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17d33109b8a64610aef8b2a48e8af68b : SUCCESS in 5m 15s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/8cc70609d1e5433c86728fd4f87052f2 : SUCCESS in 9m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e99a4bd86e094ae796c4bbb6e98fb106 : FAILURE in 14m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a3e3837d52c4b5b9f19a111764f92b9 : FAILURE in 23m 48s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7505d9775c11440f83c88f233e7167d2 : SUCCESS in 54m 00s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/262dfb2119d7488cb8f0c9d304f0a08d : SUCCESS in 6m 44s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c856fedfb738469f8be1130ec9c48625 : FAILURE in 13m 47s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7d87d91fb87c41e39cc648adc301e45e : SUCCESS in 1h 09m 35s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/f9264d8568e644b6b6ef6b1f74401250 : FAILURE in 12m 38s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/518551fbdf564c31bf4a784326ba8df7 : SUCCESS in 1h 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"61bba67e4f561caedb1f40b9ac6f7ae3abe6f627","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-14 12:12:49.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/7/check/tempest-dsvm-full-bionic-py3/4e4732c/ : SUCCESS in 1h 46m 36s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/7/check/devstack-plugin-ceph-dsvm-bionic-py3/faf8831/ : FAILURE in 1h 23m 49s","accounts_in_message":[],"_revision_number":7},{"id":"dc10c1ef17722ba6041214a1f65278d0ab0bdfda","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-14 15:21:22.000000000","message":"Patch Set 7:\n\n@Elod: i am ok with merging this if you can get it to work ... i just got thoroughly frustrated with having the l-c apparently working in my local rocky environment and then breaking in zuul that i just gave up and decided to rip the whole thing out.","accounts_in_message":[],"_revision_number":7},{"id":"db8fc2ce1d6f655cc03540c1c3f416abf31a29dd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-14 17:22:46.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e @Elod: i am ok with merging this if you can get it to work ... i just got thoroughly frustrated with having the l-c apparently working in my local rocky environment and then breaking in zuul that i just gave up and decided to rip the whole thing out.\n\nThanks Brian! Yeah, I know that frustration... :S\nAnyway, if pypi let us merge this patch then hopefully the problem will disappear.","accounts_in_message":[],"_revision_number":7},{"id":"06809caad0de00d13d25562864a8843d72b577d5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-14 17:25:48.000000000","message":"Patch Set 7:\n\nrecheck - distutils.errors.DistutilsError: Could not find suitable distribution for Requirement.parse(\u0027pbr\u0027...\n\nThis issue is now quite frequently hits xenial based jobs. According to my understanding this is some pypi mirror fallback (?) issue (missing metadata @ mirror for packages).","accounts_in_message":[],"_revision_number":7},{"id":"07c31be004cb523f39cfdd8abeca4d4c1eddd079","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-14 17:54:47.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/7/check/dsvm-tempest-cisco-zonemanager-job/37458a4/ : FAILURE in 23m 04s","accounts_in_message":[],"_revision_number":7},{"id":"09621cfe869d13227b82ff96c4e12e706bef38cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 19:11:37.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260d66046310478e8a7f3170246314b9 : SUCCESS in 9m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fecb6ac8885149308b5e804bd39e39ba : SUCCESS in 10m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f21ebb2b95c94780856c0b78215d24ad : SUCCESS in 10m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e4f005b5002a478096402d7af943d551 : SUCCESS in 8m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c18a010f96b9439db1114fb660bcc699 : SUCCESS in 3m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2ff5790cf13d481086c3df35b90add1d : SUCCESS in 1h 07m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ab1dc40e62664e11877f6f5b44c16848 : FAILURE in 11m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eada0543deed411b8aa926a80ccd5a74 : SUCCESS in 57m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af7cbf935f3c41fbad662a2c05a4b475 : SUCCESS in 7m 20s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5b4cd1e393e74fe0b94988eb4666482c : FAILURE in 14m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3f6d83109e04e378baa0edbbaeeeaf3 : SUCCESS in 8m 34s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/8de711a150234d208eb57bbcfe128a54 : SUCCESS in 44m 52s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5f852910e6a94747a6542c9b9e6c9e3e : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b512080fcb24b18b9b485e1ee8147bb : SUCCESS in 12m 14s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77d08eccf2ea4abb9604d45690366774 : SUCCESS in 8m 22s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/9227c95a96554fefa923f2b500c7e532 : SUCCESS in 10m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b2eba6f6d59c48b29b786b972ab5da54 : FAILURE in 16m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ddf5f8a2c9748349bdf47ca9947a061 : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b16de7d1d122406181482142e4df314e : SUCCESS in 7m 37s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0419e12bbf004a0dbaeff87c76da17af : FAILURE in 13m 01s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/81ffd624e04e4b8c9930c000fa9b39c6 : SUCCESS in 1h 11m 19s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/6a9ff92dbe964f3a9ae95a31d7bc1497 : FAILURE in 15m 38s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/00bf7c5ae7ba4510a4dab74b2fdf4db1 : SUCCESS in 1h 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"842d3160e6a12d083ce03794c94f1c4593addda2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-14 22:34:24.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/7/check/tempest-dsvm-full-bionic-py3/52abf6e/ : SUCCESS in 1h 43m 33s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/7/check/devstack-plugin-ceph-dsvm-bionic-py3/5fcccee/ : FAILURE in 1h 26m 51s","accounts_in_message":[],"_revision_number":7},{"id":"d7e081606049a6433f0dd19e14973c9457eaf967","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-04-15 01:05:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/06/767106/7/20210415080001/NEC-ISCSI : SUCCESS in 2h 02m 02s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/06/767106/7/20210415080001/NEC-FC : SUCCESS in 2h 05m 41s","accounts_in_message":[],"_revision_number":7},{"id":"3be751220852f8261880e2fb3d5646699faf8b68","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-20 18:23:40.000000000","message":"Patch Set 7:\n\nSame errors Elod mentioned earlier:\n\n  Couldn\u0027t find index page for \u0027pbr\u0027 (maybe misspelled?)\n  No local packages or download links found for pbr\u003e\u003d1.8","accounts_in_message":[],"_revision_number":7},{"id":"2a70dfbde26e3380935e0c3cd547bc8b1c6a04d3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-20 18:24:42.000000000","message":"Patch Set 7:\n\nrecheck neutron-grenade - it\u0027s been a week, maybe the mirrors have stabilized","accounts_in_message":[],"_revision_number":7},{"id":"62e4ed7bb9328ca80645cc7b2cc49173b3455c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 19:53:59.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e071a4a9c3324aacbadad5ebf8e0e031 : SUCCESS in 8m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91e07ecacbc74dd69c038d4decd80be5 : SUCCESS in 10m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6631e62daf884d76beebe6e41f5429aa : SUCCESS in 12m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7eeb620037e6440fb8c3e6f9538b605a : SUCCESS in 8m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f0388ae7a0245afbc4782abdb7497ab : SUCCESS in 5m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/61e5a7c12101484f979395ea7cced547 : SUCCESS in 1h 16m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3ae23b01b90a466f9a76bfef7790f289 : FAILURE in 9m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d03ec73f68cf43e8a82833923f363dd0 : SUCCESS in 1h 22m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da86d201e08349849fcdfc1f233d5c32 : SUCCESS in 8m 02s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec36f1ef9b564a8492f3ec14b0aa1430 : FAILURE in 12m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4cb2ffb5bdce4a8f84bb12d751837397 : SUCCESS in 7m 54s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/cb6eaf50fee14b6099a4375c3f2a468e : SUCCESS in 55m 23s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/64757067212f4ba7bf87fcf7e67bde94 : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b1659f3dbb84af3a4ad29136a1aee79 : SUCCESS in 11m 18s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6baf5b8fbed34e0e9f08b9df721fb8b2 : SUCCESS in 9m 17s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f20d1e3dc14e699ca014c17c27428e : SUCCESS in 8m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e45fb481508846869320d6e77f89acc8 : FAILURE in 16m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d572efa574d474baed6abe82c605974 : FAILURE in 22m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ce61669298174971a4bfc92f098c2acf : SUCCESS in 7m 18s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0152a196c7c74414bde3386fafa6fb1b : FAILURE in 12m 53s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/747b44ee45a1443b8a1193081ea881bb : SUCCESS in 1h 16m 30s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/eb3021ad12034305a0e15b595c298141 : FAILURE in 12m 59s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/d3348be1f0d44bc5bddb7ed7c622efac : SUCCESS in 1h 16m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4e0884647f6323fe3dd1f7ad98d7fecfaeb2a795","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-20 20:54:35.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e Same errors Elod mentioned earlier:\n\u003e \n\u003e   Couldn\u0027t find index page for \u0027pbr\u0027 (maybe misspelled?)\n\u003e   No local packages or download links found for pbr\u003e\u003d1.8\n\nActually, I got the hint that maybe this is related to this:\n\nhttps://github.com/pypa/pypi-support/issues/978\n\nwhich is present in some Xenial based jobs (but not everywhere, for example I haven\u0027t seen this in nova). So this needs some work and I\u0027ll try to come up with some workaround... Otherwise Rocky (and older branches - I know cinder is not really hit by older branches o:)) will become completely blocked soon.\n\nSorry for not commenting it sooner. :/","accounts_in_message":[],"_revision_number":7},{"id":"6a335d51e068ce16d2dc9fc94bac941684dde136","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-21 02:15:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/7/check/dsvm-tempest-cisco-zonemanager-job/bf9cfdb/ : FAILURE in 21m 45s","accounts_in_message":[],"_revision_number":7},{"id":"4dd699218dff37ab25ac2a6857667447f1cecb09","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-21 05:05:25.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/7/check/tempest-dsvm-full-bionic-py3/ab8e691/ : SUCCESS in 1h 22m 14s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/7/check/devstack-plugin-ceph-dsvm-bionic-py3/a7b77a9/ : FAILURE in 1h 25m 42s","accounts_in_message":[],"_revision_number":7},{"id":"3427d17fb9bc32514f12e84097857b21b3607cac","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-21 15:34:26.000000000","message":"Patch Set 7:\n\nrecheck - neutron-grenade is fixed in rocky ( https://review.opendev.org/c/openstack/devstack/+/791246 )","accounts_in_message":[],"_revision_number":7},{"id":"48f1628db7c2667ce76fb3bbdd993faf622f3a45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 17:05:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e17a8b25dd2465394e50c599e038ae7 : SUCCESS in 11m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41e250c79a2146aab8feabfb438a34d9 : SUCCESS in 11m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a2f5235ddce341a097173502bfe38fe7 : SUCCESS in 11m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/070a8a8721cc46978615cab315be4297 : SUCCESS in 11m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/47c20ebf4d5c4a8291d3fa17f2da6006 : RETRY_LIMIT in 7m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a2b0a244154c4e028380c5a9797f6ab7 : SUCCESS in 1h 29m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/856bd965963c4501bf63bc05f82384b1 : SUCCESS in 54m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45af29679a17449eab0cd08620bf915a : SUCCESS in 1h 22m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0f51bd6c7c94961a717fa5b4ec21da9 : SUCCESS in 7m 28s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d2e8cef43d94c4abaa6507bd97d70c5 : FAILURE in 10m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2236e7146c0c4bd7848e02122208c55b : SUCCESS in 8m 54s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/a94e015e529341bd8dec6a48ea0796f9 : FAILURE in 1h 09m 58s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c1cd450bfdd24a3dbe68eb477afb9ee9 : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3c88ac5afe4d9c8a66749bbdea8e06 : SUCCESS in 14m 53s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c827cb545144a7aa79b4f86dc5342f2 : SUCCESS in 8m 26s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/1ab5dd4bfedb4e6ebaaf741c07f173e3 : SUCCESS in 10m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25ec255f17a64eb0aa37dd077dde00f5 : FAILURE in 17m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f132bf6dadcc45da9563627a300831fb : FAILURE in 30m 26s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0d1589e7910b4869a02aaf32b3f00b2d : SUCCESS in 8m 28s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce2b5afb94394c488af4184e715b4bbd : SUCCESS in 1h 00m 57s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61a7fb0d379846edbfdac807c665a84e : FAILURE in 1h 16m 28s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/1d69a7016dce43da80c5b6a01c5dcbf1 : FAILURE in 19m 39s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/96100dc23ff54ee193a70127cf6d255a : SUCCESS in 1h 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9c88cb9dc59c2d8f4db313329dab4c5110b21ee9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-21 17:20:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/7/check/dsvm-tempest-cisco-zonemanager-job/94d5609/ : FAILURE in 23m 29s","accounts_in_message":[],"_revision_number":7},{"id":"50a4fe1f6bb5fe8ae46ac1b30b22704b54446e09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-21 18:51:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6655d5eb79482e6a5444e826fa8aff8bce7ca59e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-21 19:06:21.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/06/767106/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ebd4529/ : RETRY_LIMIT in 3m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/06/767106/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b643509/ : RETRY_LIMIT in 3m 50s","accounts_in_message":[],"_revision_number":8},{"id":"a5e6ec8dfaa0984d32b8b86803e0064acd1a602f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-05-21 19:19:49.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/8/check/dsvm-tempest-cisco-zonemanager-job/5bc8455/ : FAILURE in 22m 44s","accounts_in_message":[],"_revision_number":8},{"id":"dbf8de2824f4ad4044d72b3c11ecaf30219d651b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 19:27:57.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/06/767106/8/check/iscsi-driver-nimble/0881aa5/0881aa5.html : FAILURE in 26m 55s","accounts_in_message":[],"_revision_number":8},{"id":"1ad57c9592cba559c1c709663f49205e5d8301d7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 19:43:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/06/767106/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/08bb71d/08bb71d.html : FAILURE in 32m 25s","accounts_in_message":[],"_revision_number":8},{"id":"5205c02ab36a5616d8699ae3123f8c73efb89363","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-05-21 19:45:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/06/767106/8/check/iscsi-driver-primera/d18cb40/d18cb40.html : FAILURE in 33m 30s","accounts_in_message":[],"_revision_number":8},{"id":"01b83b30828d5feb38df287d97b442a5f908f35b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 20:31:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d531920fba9a4912817345c7e639a2b6 : SUCCESS in 9m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/278f0fe945164a34845542869652e78f : SUCCESS in 12m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/efdb2c9a6e6b4071998f37d6c275cbce : SUCCESS in 10m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5fc64a8bb3f14d08bdb64036c08a3e45 : SUCCESS in 10m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a38736ba24344735ba1d4c9433aaec15 : RETRY_LIMIT in 3m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4d6f84b2344f4f038bf2538b2bc66591 : SUCCESS in 1h 13m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ac009224a82b46528184f4e159a52496 : SUCCESS in 39m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8ba4dd74a5a4d0d8a3626bec754c31f : SUCCESS in 59m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c1f6ef2b31b43ae90acc3b10251665c : SUCCESS in 6m 03s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2c025f4d0fb4446aae0ad5de8ed62c1d : SUCCESS in 1h 16m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03f84c61abae4d46856ee5c4a6b0fd4c : SUCCESS in 9m 02s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/6abc7c89d20849a2bc6d1296169de235 : SUCCESS in 53m 28s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7ef144dac16649229d1f47efc1d1f07a : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42f0cc86dc2a49469c53e55f3ab9bc44 : SUCCESS in 10m 36s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a00dce6205a495dbf9df35436ce0fb7 : SUCCESS in 5m 52s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/088020f9f5fe4a6f905faf3acce08ac0 : SUCCESS in 10m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1bc4313ef96d4a9cbd0f0d559f79b79f : FAILURE in 11m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d3b40c3ab4ee4834b91318e62c0b48b1 : SUCCESS in 33m 19s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ab6331f3d8b34eac98f5b75c7b641383 : SUCCESS in 6m 08s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/926917f51a2242d785e8880406f4f5d3 : SUCCESS in 1h 05m 56s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cd94d7eb84ef48f6969b035926f043fc : SUCCESS in 51m 58s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/3f43321d4f6c4a89a7d9e311e991b8ae : FAILURE in 6m 55s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/6fe6e4de92194b619cf400cc4f97c3b7 : SUCCESS in 1h 34m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"16b0c1b986fbdef1838e2d753b4e41898510d4d2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-21 21:01:23.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/7b7c82ab423b46cab5a4d43b9e2982bb : POST_FAILURE in 1h 32m 38s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/07fec6a8de674a4d960a76b56fd506d6 : POST_FAILURE in 1h 09m 52s","accounts_in_message":[],"_revision_number":8},{"id":"9f839cacdbcf37a81b90840940511cfbd52cff41","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-02 15:38:25.000000000","message":"Patch Set 8:\n\nrecheck - I don\u0027t know why test-setup.sh failed with SQL syntax (in requirements-check job), as it worked before...","accounts_in_message":[],"_revision_number":8},{"id":"227063c118ed209d300305bd1e7e4c8c8fb215ab","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-02 16:07:32.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/8/check/dsvm-tempest-cisco-zonemanager-job/ba00caf/ : FAILURE in 23m 22s","accounts_in_message":[],"_revision_number":8},{"id":"86094192cf0e96f3983107897a473185884d19fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 19:18:14.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f8f72ff32f402eb8ae45567ff74a9c : SUCCESS in 9m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b39d7616b11a4265b051d78ee8ff22d8 : SUCCESS in 10m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/549cd11e73f14800826eb60b63d30a81 : FAILURE in 10m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/474ba04b2d5f45d3b3d7bfdd0eecf6fb : SUCCESS in 10m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/308b406db54b4feb898b5808670d9947 : RETRY_LIMIT in 3m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c287aa69cd9d4ec3a722e86e94aae063 : SUCCESS in 1h 26m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a37dbda635fc43749b00c42e6c69583a : SUCCESS in 1h 00m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63379b3632b04305a26fa171b63d7749 : SUCCESS in 1h 29m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aec4991af32496fa40c22f70cd59a6e : SUCCESS in 7m 01s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/77d82211147348beb3f16b978e6efdb3 : SUCCESS in 1h 49m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1bbc2901caca434c8250a84fea97224e : SUCCESS in 9m 29s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/56231f8f0f834c848c544aec388d0c64 : SUCCESS in 1h 03m 23s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4b9aa391aa2d411286fc2c9de4e5d24e : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea7a9ea6f099478d91ca789a78aaada6 : SUCCESS in 13m 46s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6eadd26255894df8b60936b3fbb4e85f : SUCCESS in 7m 06s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/2219485daf1342629cee75a6b8ad6292 : SUCCESS in 10m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cc4cfb846c9e4b718681bddd623485d0 : FAILURE in 21m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a0d48c52acd4b70a1f9886b92d61832 : SUCCESS in 31m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0a46fca42404450c91eb9368703a147c : SUCCESS in 7m 46s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e1e7bcc173f4d2ea7f9168b33b1b0a0 : SUCCESS in 1h 13m 32s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0ef9677be8574247abfac38ee3cecae4 : SUCCESS in 1h 30m 34s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/ea2c97dec9d441e4a6931cbb3c8ec6dd : FAILURE in 16m 46s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/607e7fd8423b4dd187db3b27ae687d2f : SUCCESS in 1h 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5ee6a87d7b2438373b30d006ee76b8de7b89baad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-02 19:57:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c172c340d4835d0a6025af28c7478d79517fe192","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-02 20:02:23.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"1cb1c2efd2969205696592eed8e71273285ee026","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-06-02 20:05:04.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"34ff91dd111eada99d4ddb0f7ebfd90bc74fac44","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-06-02 20:25:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/06/767106/9/check/dsvm-tempest-cisco-zonemanager-job/416b0d2/ : FAILURE in 22m 33s","accounts_in_message":[],"_revision_number":9},{"id":"6e02eb023d2ba0d1277a8ab67b9ad474c0d27da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 21:43:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159f89bbf70f4abfafd74555b88a71b3 : SUCCESS in 10m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1167f24c71b4e3e8553119b4089170a : SUCCESS in 9m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dbe0fabd6a9a4b3c92545fd2de99a6e7 : SUCCESS in 11m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3e21b18f7db74645b4f48a39eb82662a : SUCCESS in 10m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5df0130d5c6e44479766d0001db47041 : SUCCESS in 5m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/40871deaeeb144c1b616bd64e2446fdb : SUCCESS in 1h 33m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6493ede80e3d4ed9aed8d5a5b1345249 : SUCCESS in 1h 01m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15fc57ecaa514b7b9e1161883aac29e5 : SUCCESS in 1h 07m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/800736918bdb483eafafe6b1c42a8ac4 : SUCCESS in 6m 51s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf061a7cc16245129cf536f4478a07d1 : SUCCESS in 1h 42m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e19393f72c2426d98e23bd74ac0e7b6 : SUCCESS in 9m 19s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/a3b3df00abd04cacbb0465eef0de099f : SUCCESS in 1h 00m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/954b3f34cd8d48548d21dd9159ba04d9 : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0574684bc9e4015b2e8e6c0fad18b76 : SUCCESS in 13m 53s\n- cinder-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2e2d89bae42450b92f91807cc219a5b : SUCCESS in 11m 11s\n- cinder-tox-py36 https://zuul.opendev.org/t/openstack/build/b9bda795bfcf49b2a54d086a28963b03 : SUCCESS in 10m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bcb0cb21a7ff4c2a929786b3cb8518fd : FAILURE in 11m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f132e96e52c49c9a168bf7a210eefec : SUCCESS in 30m 16s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5c12b879ec5c471199412e2e86f72b7a : SUCCESS in 6m 01s (non-voting)\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f98fa24849e14704a1551b4258b565c6 : SUCCESS in 50m 36s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c5746ebec2db4545906347aea65178d0 : SUCCESS in 1h 13m 23s (non-voting)\n- legacy-tempest-dsvm-full-drbd-devstack https://zuul.opendev.org/t/openstack/build/fb8e40980f634efc839288251f45bcc7 : FAILURE in 18m 16s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-nfs https://zuul.opendev.org/t/openstack/build/7001eb42e0494f3c8757a7c7c5bad46e : SUCCESS in 1h 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"692f47361ca681e157a55eb0defc24dd3d58aa16","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-02 21:44:39.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/06/767106/9/upstream-solidfire-check/cinder-solidfire-iSCSI/d24cf40/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"8a1b3b59716b215e49c11d40fc1ab4a4d1cd6baa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-02 21:53:24.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/06/767106/9/upstream-check/cinder-cDOT-iSCSI/95bc5ad/ : FAILURE in 9m 08s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/06/767106/9/upstream-check/cinder-cDOT-NFS/291a4cb/ : FAILURE in 8m 23s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/06/767106/9/upstream-check/cinder-cDOT-FCP/a35cbef/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"6689c5e57d566ef499e2dd6fcece9c2e9b7189a1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-03 00:24:15.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/9/check/tempest-dsvm-full-bionic-py3/929e04e/ : SUCCESS in 1h 37m 50s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/06/767106/9/check/devstack-plugin-ceph-dsvm-bionic-py3/5f7687a/ : FAILURE in 1h 14m 06s","accounts_in_message":[],"_revision_number":9},{"id":"a961e8ae17806ad0d427ed5454a29ff177acbd21","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-03 12:15:59.000000000","message":"Patch Set 9: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"55aa69646126cc6c298455a65ff5f8b20b7b4f8f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-04 15:33:36.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nLooks good to me.  Thanks for sticking with this, Elod!","accounts_in_message":[],"_revision_number":9},{"id":"b8a630271f5e1a52ad9837bbcd67533ddff3807c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 15:34:35.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"c9013075b96abb042a81b3eeccfddb961d9cbb8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:16:51.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfdcb7f47d6f4861b491b3ed46ccdf99 : SUCCESS in 13m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8a7a7964140430eac79e0fc1d20a682 : SUCCESS in 10m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/636318d27cd441949c1d15cf98083a64 : SUCCESS in 11m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1bb0c2ef0961468b8b6f250e3a8b1909 : SUCCESS in 9m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb9b1b1f10494d0f8001903852f02e09 : SUCCESS in 5m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/42827ef8867c47abb45bfe60baff3593 : SUCCESS in 1h 18m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/77947d1278cd4f969c66f87a6f1c7383 : SUCCESS in 56m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98108e4aa7f94284a39bbf1dd9bc928a : SUCCESS in 1h 19m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/861c57a64a4e442384ff118f9a9ebd7e : SUCCESS in 11m 10s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/b816c83e3bef4a23bd48bb02ee811922 : SUCCESS in 1h 08m 51s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/bcbb8834ec3f4b0a9b8f35c3479ce2ef : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/935a8cf553894030bb1335225ecb4196 : SUCCESS in 14m 19s\n- legacy-grenade-dsvm-cinder-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3b0cc214c96a493d8a8c0db854f412bb : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":9},{"id":"a5b42b088bd9f378a8bb52589967139fedffd8ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:16:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"4a18eed01e439406bdcbc3bd002367daaaf6fc36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:20:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e05cb29119ef41ff8eff9ddb3f1db107 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":9},{"id":"2e85efce9a59360cf45b8e1fcff9a7ea0325fe93","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-06-08 20:46:34.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/06/767106/9/ : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"62fc927e5af452bc8ddf32bca9685a37194ce019","revisions":{"026f9590caf579c2579a7c0370abea2496fbc818":{"kind":"REWORK","_number":1,"created":"2020-12-15 09:39:25.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/06/767106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/1"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:38:56.000000000","tz":120},"subject":"Pin bandit to v1.1.0","message":"Pin bandit to v1.1.0\n\nWe don\u0027t have test requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7.\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit 4add6a7631f89fe62d0cee8a167f05894d33a619)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/026f9590caf579c2579a7c0370abea2496fbc818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/026f9590caf579c2579a7c0370abea2496fbc818"}]},"branch":"refs/heads/stable/rocky"},"245c4401081a36f7a835f532155068c4d96b812d":{"kind":"REWORK","_number":2,"created":"2020-12-15 09:47:00.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/06/767106/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/2"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:46:23.000000000","tz":120},"subject":"Cap bandit to v1.6.3","message":"Cap bandit to v1.6.3\n\nWe don\u0027t have test requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7.\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit fc23771b7528c2e25f414e8584bd43c4b3e36184)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/245c4401081a36f7a835f532155068c4d96b812d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/245c4401081a36f7a835f532155068c4d96b812d"}]},"branch":"refs/heads/stable/rocky"},"312e81177db4580f388f61574407ce62af366b7f":{"kind":"REWORK","_number":3,"created":"2020-12-15 10:02:30.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/06/767106/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/3"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 10:02:11.000000000","tz":120},"subject":"Cap bandit to v1.6.2","message":"Cap bandit to v1.6.2\n\nWe don\u0027t have test requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7.\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit 270065e0d42f6c07e9d3b1e875950e4439f7c228)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/312e81177db4580f388f61574407ce62af366b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/312e81177db4580f388f61574407ce62af366b7f"}]},"branch":"refs/heads/stable/rocky"},"7f77de0519d8d63cea47fcf84831cde625380411":{"kind":"REWORK","_number":4,"created":"2020-12-15 10:34:33.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/06/767106/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/4"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 10:34:12.000000000","tz":120},"subject":"Cap bandit to v1.6.2","message":"Cap bandit to v1.6.2\n\nWe don\u0027t have test requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7.\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit a512bfc233fc76c4375388b8546ab9e60459e1d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f77de0519d8d63cea47fcf84831cde625380411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f77de0519d8d63cea47fcf84831cde625380411"}]},"branch":"refs/heads/stable/rocky"},"d475a3e1380e7df97123334e1a48a2d525c742fb":{"kind":"REWORK","_number":5,"created":"2021-04-13 17:21:45.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/767106/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/5"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-04-13 17:20:20.000000000","tz":120},"subject":"Cap bandit to v1.6.2 and fix constraints","message":"Cap bandit to v1.6.2 and fix constraints\n\nWe don\u0027t have bandit requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7, as\nbandit 1.6.3 [1] release has dropped support for py2 [2] but the\nrelease is faulty and pip still picks it up for py2 [3][4].\n\nWith the new pip resolver the following issues needed to be fixed:\n* SQLAlchemy-Utils constraint conflicts with rocky\u0027s upper constraints\n* bump lower constraint of cryptography\n* bump lower constraint of oslo.messaging\n* bump lower constraint of oslo.utils\n\ninstall_command needs to be added to lower-constraints tox target to\nwork properly.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit a512bfc233fc76c4375388b8546ab9e60459e1d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d475a3e1380e7df97123334e1a48a2d525c742fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d475a3e1380e7df97123334e1a48a2d525c742fb"}]},"branch":"refs/heads/stable/rocky"},"bfcd51959323898a6689942f14aaaab212d962dd":{"kind":"REWORK","_number":6,"created":"2021-04-13 18:18:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/767106/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/6"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-04-13 18:16:21.000000000","tz":120},"subject":"Cap bandit to v1.6.2 and fix constraints","message":"Cap bandit to v1.6.2 and fix constraints\n\nWe don\u0027t have bandit requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7, as\nbandit 1.6.3 [1] release has dropped support for py2 [2] but the\nrelease is faulty and pip still picks it up for py2 [3][4].\n\nWith the new pip resolver the following issues needed to be fixed:\n* SQLAlchemy-Utils constraint conflicts with rocky\u0027s upper constraints\n* bump lower constraint of cryptography\n* bump lower constraint of oslo.messaging\n* bump lower constraint of oslo.utils\n\nIn lower-constraints job lxml requires libxml2-dev and libxslt1-dev\npackages, so bindep.txt is extended with them.\n\ninstall_command needs to be added to lower-constraints tox target to\nwork properly.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit a512bfc233fc76c4375388b8546ab9e60459e1d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bfcd51959323898a6689942f14aaaab212d962dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bfcd51959323898a6689942f14aaaab212d962dd"}]},"branch":"refs/heads/stable/rocky"},"5b55adb9ed92788a03eb0cd0ccdd11b4cf1a8ea6":{"kind":"REWORK","_number":7,"created":"2021-04-14 09:49:38.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/767106/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/7"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-04-14 09:42:32.000000000","tz":120},"subject":"[stable-only] Cap bandit to v1.6.2 and fix constraints","message":"[stable-only] Cap bandit to v1.6.2 and fix constraints\n\nThis patch has multiple fixes to unblock gate.\n\n1. bandit\n\nWe don\u0027t have bandit requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7, as\nbandit 1.6.3 [1] release has dropped support for py2 [2] but the\nrelease is faulty and pip still picks it up for py2 [3][4].\n\n2. pip resolver\n\nWith the new pip resolver the following issues needed to be fixed:\n* SQLAlchemy-Utils constraint conflicts with rocky\u0027s upper constraints\n* bump lower constraint of cryptography\n* bump lower constraint of oslo.messaging\n* bump lower constraint of oslo.utils\n\n3. lower-constraints\n\nIn lower-constraints job lxml requires libxml2-dev and libxslt1-dev\npackages, so bindep.txt is extended with them.\n\ninstall_command needs to be added to lower-constraints tox target to\nwork properly.\n\n4. cinder-tempest-dsvm-lvm-lio-barbican\n\nFix for overlapping CIDR issue in cinder-tempest-dsvm-lvm-lio-barbican\nis the same as in patch I068cf1e9618d305b5a9383c283bfa0f120bfe905.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit a512bfc233fc76c4375388b8546ab9e60459e1d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b55adb9ed92788a03eb0cd0ccdd11b4cf1a8ea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b55adb9ed92788a03eb0cd0ccdd11b4cf1a8ea6"}]},"branch":"refs/heads/stable/rocky"},"bc6dfbced86a5e51d243b9cc2ee32944b63a7c8a":{"kind":"REWORK","_number":8,"created":"2021-05-21 18:51:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/767106/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/8"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-05-21 18:23:50.000000000","tz":120},"subject":"[stable-only] Cap bandit to v1.6.2 and fix constraints","message":"[stable-only] Cap bandit to v1.6.2 and fix constraints\n\nThis patch has multiple fixes to unblock gate.\n\n1. bandit\n\nWe don\u0027t have bandit requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7, as\nbandit 1.6.3 [1] release has dropped support for py2 [2] but the\nrelease is faulty and pip still picks it up for py2 [3][4].\n\n2. pip resolver\n\nWith the new pip resolver the following issues needed to be fixed:\n* SQLAlchemy-Utils constraint conflicts with rocky\u0027s upper constraints\n* bump lower constraint of cryptography\n* bump lower constraint of oslo.messaging\n* bump lower constraint of oslo.utils\n\n3. lower-constraints\n\nIn lower-constraints job lxml requires libxml2-dev and libxslt1-dev\npackages, so bindep.txt is extended with them.\n\ninstall_command needs to be added to lower-constraints tox target to\nwork properly.\n\n4. cinder-tempest-dsvm-lvm-lio-barbican\n\nFix for overlapping CIDR issue in cinder-tempest-dsvm-lvm-lio-barbican\nis the same as in patch I068cf1e9618d305b5a9383c283bfa0f120bfe905.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit a512bfc233fc76c4375388b8546ab9e60459e1d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc6dfbced86a5e51d243b9cc2ee32944b63a7c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc6dfbced86a5e51d243b9cc2ee32944b63a7c8a"}]},"branch":"refs/heads/stable/rocky"},"62fc927e5af452bc8ddf32bca9685a37194ce019":{"kind":"REWORK","_number":9,"created":"2021-06-02 19:57:06.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/767106/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/06/767106/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/06/767106/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/06/767106/9"}}},"commit":{"parents":[{"commit":"91619ce3297553401c01c371690b29b68f11f17f","subject":"PowerMax Driver - Legacy volumes fail to live migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91619ce3297553401c01c371690b29b68f11f17f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-02 19:48:59.000000000","tz":120},"subject":"[stable-only] Cap bandit to v1.6.2 and fix constraints","message":"[stable-only] Cap bandit to v1.6.2 and fix constraints\n\nThis patch has multiple fixes to unblock gate.\n\n1. bandit\n\nWe don\u0027t have bandit requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7, as\nbandit 1.6.3 [1] release has dropped support for py2 [2] but the\nrelease is faulty and pip still picks it up for py2 [3][4].\n\n2. pip resolver\n\nWith the new pip resolver the following issues needed to be fixed:\n* SQLAlchemy-Utils constraint conflicts with rocky\u0027s upper constraints\n* bump lower constraint of cryptography\n* bump lower constraint of oslo.messaging\n* bump lower constraint of oslo.utils\n\n3. lower-constraints\n\nIn lower-constraints job lxml requires libxml2-dev and libxslt1-dev\npackages, so bindep.txt is extended with them.\n\ninstall_command needs to be added to lower-constraints tox target to\nwork properly.\n\n4. cinder-tempest-dsvm-lvm-lio-barbican\n\nFix for overlapping CIDR issue in cinder-tempest-dsvm-lvm-lio-barbican\nis the same as in patch I068cf1e9618d305b5a9383c283bfa0f120bfe905.\n\n5. pin nodeset to xenial for requirements-check job\n\nThe default nodeset for zuul jobs is now ubuntu-focal, and\nrequirements-check job fails on focal, because the mysql it contains is\nnot compatible with the commands that is used in tools/test-setup.sh.\nThis patch pins the nodeset to ubuntu-xenial for requirements-check job\nto make it pass.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit a512bfc233fc76c4375388b8546ab9e60459e1d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62fc927e5af452bc8ddf32bca9685a37194ce019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62fc927e5af452bc8ddf32bca9685a37194ce019"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
