)]}'
{"id":"openstack%2Fcinder~767107","triplet_id":"openstack%2Fcinder~stable%2Fqueens~Ie597e778e3efa61bfd98eaaa92730b5050195e7a","project":"openstack/cinder","branch":"stable/queens","topic":"py27-requirements","hashtags":[],"change_id":"Ie597e778e3efa61bfd98eaaa92730b5050195e7a","subject":"Cap bandit to v1.6.2","status":"MERGED","created":"2020-12-15 09:40:48.000000000","updated":"2021-01-16 01:14:42.000000000","submitted":"2021-01-16 01:12:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767107-py27-requirements","meta_rev_id":"c95007b7f013d7c40aa3ec89cbc6c3c297e980aa","_number":767107,"virtual_id_number":767107,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-16 01:12:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-07 13:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-01-07 14:00:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2021-01-13 18:56:41.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-01-07 14:00:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 10:54:04.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-15 10:57:39.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-15 11:09:33.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 12:45:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 17:42:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 02:19:15.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 02:23:41.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-01-07 13:59:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-07 13:59:56.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2021-01-07 14:00:06.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-01-13 20:28:14.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-15 11:02:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"e5088e18f72c0873881996dd05489b7b65340564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 09:40:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e737f644aee482482f35a3359e23303e0fbc6ab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 09:48:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8b8484e001cea98b5758e2c05b37f9b61350f84","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 09:51:17.000000000","message":"Topic set to py27-requirements","accounts_in_message":[],"_revision_number":2},{"id":"106e995922930aa3902372ebe79d12357b852876","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 10:04:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ed1a64a33e4d181d115b8c44173556b679dc0e8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 10:35:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"11cbf38ce0e520887b6c8c15926571753cfcc63f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-15 10:54:04.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767107/4/iscsi/ : FAILURE in 8m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/767107/4/smb/ : FAILURE in 9m 13s","accounts_in_message":[],"_revision_number":4},{"id":"d2d9b57a4136cc267887309aefdd9a70cbecffe9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-15 10:57:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/767107/4/check/dsvm-tempest-cisco-zonemanager-job/1f1299c/ : FAILURE in 16m 44s","accounts_in_message":[],"_revision_number":4},{"id":"2fb6a31df8995313444f3fcc8f21711e91088124","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-15 11:09:33.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/767107/3/check/tempest-dsvm-full-xenial/006ec62/ : FAILURE in 23m 42s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/767107/3/check/tempest-dsvm-full-xenial-py3/c46a5cc/ : FAILURE in 25m 54s","accounts_in_message":[],"_revision_number":3},{"id":"04037f43d8f76559c9854b3808b1212e4d6c3e35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 12:45:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc342b8b0e444088c2122de65dcaf19 : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6e76561b79c40fe976df7ef6c03e4b9 : SUCCESS in 9m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/10014fb5af744e2fa3c5cf07ab420b36 : SUCCESS in 19m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c3fe45754d184b628c30409c3fd7c90e : SUCCESS in 28m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9aee216daa674d458784448c2a81b60d : SUCCESS in 8m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5150d9ecf83b4e67b44c4b05ac8fa9a8 : FAILURE in 2h 00m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/788e25223b48439fa1fa0746969b8d33 : SUCCESS in 1h 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6050f95d7c6f4858b4dabcf3c7ea884a : SUCCESS in 11m 27s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d31793be6304ee5bc0c0404df0356bc : FAILURE in 1h 04m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e1d3248253446f2960b57cbc595df3b : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/907ba67343b24dad9996f06f21efc9ae : FAILURE in 5m 53s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e637bacd504e44a5b1f6b64844bf4aef : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ddf1607a1d3caf255592d4e22f5c43c600319170","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 15:30:18.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"aa3351cfcdb3465334f45769f4a4cdddf651daae","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-15 15:47:32.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767107/4/iscsi/ : FAILURE in 8m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/767107/4/smb/ : FAILURE in 8m 34s","accounts_in_message":[],"_revision_number":4},{"id":"47b3ab4553e336ec7c7602b406529593ef06d47c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-15 15:51:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/767107/4/check/dsvm-tempest-cisco-zonemanager-job/985a3b4/ : FAILURE in 16m 19s","accounts_in_message":[],"_revision_number":4},{"id":"6e4e9d94a00310d49282347d1e1fec175dad9cac","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-15 17:42:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/07/767107/4/20201216003551/NEC-ISCSI : SUCCESS in 2h 06m 45s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/07/767107/4/20201216003551/NEC-FC : SUCCESS in 2h 07m 03s","accounts_in_message":[],"_revision_number":4},{"id":"b48536c4acfc51caaf7153b1adf7f0e62bff4452","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-15 21:15:13.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/767107/4/check/tempest-dsvm-full-xenial/9456582/ : FAILURE in 39m 32s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/767107/4/check/tempest-dsvm-full-xenial-py3/43e9712/ : FAILURE in 34m 02s","accounts_in_message":[],"_revision_number":4},{"id":"49da6960ab3b69e76a7ab1623fa2506f03d46d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 01:37:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53160715c9dc4947bf7821299e370e0e : SUCCESS in 8m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8050b0c4ea834bdd88e5f2abb852a9f6 : SUCCESS in 8m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/033a75f802b44fe18826a34aebebd4dd : SUCCESS in 8m 44s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ee0d250716e340229b0bdbb534f01f17 : SUCCESS in 21m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bc1587b018f04640928de8c68b367c0d : SUCCESS in 3m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bba9a3859e1a4b98a4ef27ea10eba9dc : SUCCESS in 1h 22m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80cdceff17fb47aa862bde19bb1b36bf : SUCCESS in 1h 23m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4b3d2bddd39471385fe7a1ace6021e1 : SUCCESS in 6m 48s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a646c192f15b4adeae631f03e06ba512 : FAILURE in 1h 09m 21s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cb091e6aae6849da9b65f5f4042f4d72 : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/da372955408f4bf6967f32f7b1f5ff4f : FAILURE in 11m 16s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b48cbb37d7f144248ea22a1ba1810f4c : SUCCESS in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3a4bb4478724f7ec5fa0a1adcbb57089dc74e6d1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-16 02:19:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-iSCSI/1d84b78/ : FAILURE in 25m 26s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-iSCSI-IPv6/a03f28a/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-NFS/13cd8dd/ : FAILURE in 24m 48s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-NFS-IPv6/38357d0/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-FCP/08d8440/ : FAILURE in 40m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"6701c952bd01a663506d644213edc8d6ad9b6d35","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-12-16 02:23:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/767107/4/upstream-solidfire-check/cinder-solidfire-iSCSI/1aa99ad/ : FAILURE in 44m 46s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"1b324d92b03eb6581cfd7103d065f899cfb1d291","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-07 13:59:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"623e93c9141cf0822269f5ce6543e9d6bdbde36b","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-01-07 13:59:56.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks Ivan, LGTM","accounts_in_message":[],"_revision_number":4},{"id":"d34bf8ae194d5b49a90fb48c43deb21bd0302fe6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-07 14:00:06.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nNinja-approving to unblock gate.","accounts_in_message":[],"_revision_number":4},{"id":"e42178f427819d87dad1f800a085667688816252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 14:00:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5633c77f4157aee9a850470a7c71b878b9b5a058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 16:34:29.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa14780f15f74b17884b7e359887ae88 : SUCCESS in 8m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a4d059d87434036ae6272e1d46e7be0 : SUCCESS in 9m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/232152c4efa140c4b8b6fe53af3c0aff : SUCCESS in 8m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/25fc20d6d89d44138e2656ad0a822689 : SUCCESS in 22m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4b87a8e6cb3e40f4ab74d4b031ab3609 : SUCCESS in 3m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b18c4809dc194cf788cf04ffc6941f65 : FAILURE in 1h 34m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4f437d4538b47b694157aef302ed06f : SUCCESS in 1h 18m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fa1460fc0c14ec48f5506e0196a29b8 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ddfb14bfdc757007011405b9644fbc9fb18ce11a","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-01-13 18:56:41.000000000","message":"Patch Set 4: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"524ac592cf527ba2014d4dd1d4352b9a90c3f1ba","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-13 19:05:54.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767107/4/iscsi/ : FAILURE in 7m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/767107/4/smb/ : FAILURE in 8m 20s","accounts_in_message":[],"_revision_number":4},{"id":"6a6fd1a32478a221d79936e085e9e93d723c18ae","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-13 20:28:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/767107/4/check/dsvm-tempest-zadara-driver/edb8f91 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a4b48d24437f9c90fc902d622dbd6b2389d4d070","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-13 20:35:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/767107/4/check/dsvm-tempest-cisco-zonemanager-job/fef4009/ : FAILURE in 16m 57s","accounts_in_message":[],"_revision_number":4},{"id":"01545a26ab920956d861f187e8e5ad50af1101be","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-13 20:42:03.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/767107/4/check/tempest-dsvm-full-xenial/def001c/ : FAILURE in 37m 09s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/767107/4/check/tempest-dsvm-full-xenial-py3/635ef61/ : FAILURE in 41m 27s","accounts_in_message":[],"_revision_number":4},{"id":"42913e8154c5c47971b9b85006c6f0b5f673decc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 19:58:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b85a78f072754e989be57eb913f4f20f : SUCCESS in 8m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30c07e85eca6465c8110b19e99e85475 : SUCCESS in 9m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c3511a3e1db4045959efff063bb47fc : SUCCESS in 23m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/747cc5a34b534f1d93447e3550a6aca2 : TIMED_OUT in 31m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8fa94a43909442e390fcb542caf344bf : SUCCESS in 4m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d02c02ef92524bd79117912b6b6b6abb : SUCCESS in 1h 45m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a642d12e96845e686dc8564e69b52d1 : SUCCESS in 1h 45m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98a1ea43164345eeab4c32d3ebd0a33d : SUCCESS in 15m 23s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fe1bebffd498451292ceebc4faaf06f6 : FAILURE in 19m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45a7494dd9b94016bfe40e497eed7e9d : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10b631473aa74102af5bd859c0032de6 : FAILURE in 4m 59s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/80462c0396994297bafaa92405c69dc1 : SUCCESS in 6m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"498e8a12b300e1b58b15bce926dbfb852136a030","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-15 11:02:21.000000000","message":"Patch Set 4:\n\nrecheck - doc build timeout","accounts_in_message":[],"_revision_number":4},{"id":"3675f89fef9bcfadfac3ab172fa1f432e7bc0476","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-15 11:12:06.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767107/4/iscsi/ : FAILURE in 9m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/767107/4/smb/ : FAILURE in 9m 13s","accounts_in_message":[],"_revision_number":4},{"id":"744522a0c0ab587aad68e80f736caf067ab2c102","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-15 13:05:17.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/767107/4/check/tempest-dsvm-full-xenial/bfca660/ : FAILURE in 32m 24s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/07/767107/4/check/tempest-dsvm-full-xenial-py3/0f87a7b/ : FAILURE in 33m 29s","accounts_in_message":[],"_revision_number":4},{"id":"7c67d27aa6d1914bc84a07f71067c930dd946c15","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-15 14:21:10.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/07/767107/4/check/dsvm-tempest-cisco-zonemanager-job/807da23/ : FAILURE in 17m 33s","accounts_in_message":[],"_revision_number":4},{"id":"68ef596de3204500dcde065729483f253fad2545","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-15 15:50:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/07/767107/4/check/dsvm-tempest-zadara-driver/a451a3c : SUCCESS in 1h 17m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7b124c412f069bbb033b063d34c556418da3b102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 21:56:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c20430214e464d8f3cc4bfda007b91 : SUCCESS in 9m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f9db1cd71104ea19e00da1e592589bb : SUCCESS in 10m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e9eb5cac3ac442a1ac33163235142918 : SUCCESS in 9m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1171009e85c540b28f73fd8c27422606 : SUCCESS in 17m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3e3223a4fc6465289037c96a06d9bbe : SUCCESS in 3m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8a69ca0cf74f4be1923a39c5796c5b8f : SUCCESS in 1h 19m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb2d333ed396494c9f91a6762edb9685 : SUCCESS in 1h 19m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f931924b053b44dd88641f35d556fe57 : SUCCESS in 5m 38s\n- cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7acbeef0eec74d1a942d89709a7d572b : FAILURE in 1h 06m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bd861319d83047258944a5dfe7a67423 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae962758579e4d0798a106d115cea433 : FAILURE in 4m 33s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8dd44de3484c488da506f5ed2b238554 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c47cabe7c682471b89647cc91467957270a6ac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 21:57:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1c1f5e3768a44c6caad7ed83a78930223f7d9f3e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-15 23:11:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-iSCSI/530f9a3/ : FAILURE in 25m 46s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-iSCSI-IPv6/edda3e8/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-NFS/1e9e5d2/ : FAILURE in 27m 57s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-NFS-IPv6/29d9bd1/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/767107/4/upstream-check/cinder-cDOT-FCP/5d6cba7/ : FAILURE in 1h 14m 37s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"0429cfcedb0c55c8b13ee2f35b4aa3f969f80ddb","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-15 23:16:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/767107/4/upstream-solidfire-check/cinder-solidfire-iSCSI/b88c9d8/ : FAILURE in 1h 18m 59s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"e1655ac12f4aab447a21a5461625d1962adf1e98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 01:12:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a78992fc2440f3acaeccd0cb332179 : SUCCESS in 7m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a81780b7f05c43b1b3326eaee7272806 : SUCCESS in 12m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2bfa4e9248714808a9cbf689578b2dc0 : SUCCESS in 9m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/59f7bb5ea6f244a5bb6bf8c7d45afd32 : SUCCESS in 25m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ab19657996e74c50aa65860a5573da27 : SUCCESS in 3m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e599c7459d2642a0ac2ef119d092fd87 : SUCCESS in 1h 09m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e898ef2f15bd4c6dbfd41ae7842f19eb : SUCCESS in 1h 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa28d2dc09b542288927596a59fde54c : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"efb04e6234a576e393f342532b5fd78676edf524","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 01:12:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c95007b7f013d7c40aa3ec89cbc6c3c297e980aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 01:14:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/730ea0140a3e4a11be3db1c01608d1ec : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7abe6dc063eac6f40a3c4eacde41f222989c7a36","revisions":{"e72cd65486a6c8e4f4baa450c68d4167aaf695f1":{"kind":"REWORK","_number":1,"created":"2020-12-15 09:40:48.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/07/767107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/767107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/767107/1"}}},"commit":{"parents":[{"commit":"14de07ed463fa9c9f21855b87c810ae3788e7c78","subject":"Remove experimental job legacy-tempest-dsvm-zeromq-multibackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/14de07ed463fa9c9f21855b87c810ae3788e7c78"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:40:31.000000000","tz":120},"subject":"Pin bandit to v1.1.0","message":"Pin bandit to v1.1.0\n\nWe don\u0027t have test requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7.\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit 4add6a7631f89fe62d0cee8a167f05894d33a619)\n(cherry picked from commit 026f9590caf579c2579a7c0370abea2496fbc818)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e72cd65486a6c8e4f4baa450c68d4167aaf695f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e72cd65486a6c8e4f4baa450c68d4167aaf695f1"}]},"branch":"refs/heads/stable/queens"},"3475d534bd94af619519980a1d4e59f53605bce5":{"kind":"REWORK","_number":2,"created":"2020-12-15 09:48:36.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/07/767107/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/767107/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/767107/2"}}},"commit":{"parents":[{"commit":"14de07ed463fa9c9f21855b87c810ae3788e7c78","subject":"Remove experimental job legacy-tempest-dsvm-zeromq-multibackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/14de07ed463fa9c9f21855b87c810ae3788e7c78"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:48:17.000000000","tz":120},"subject":"Cap bandit to v1.6.3","message":"Cap bandit to v1.6.3\n\nWe don\u0027t have test requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7.\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit fc23771b7528c2e25f414e8584bd43c4b3e36184)\n(cherry picked from commit 245c4401081a36f7a835f532155068c4d96b812d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3475d534bd94af619519980a1d4e59f53605bce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3475d534bd94af619519980a1d4e59f53605bce5"}]},"branch":"refs/heads/stable/queens"},"ed6ee8b26ea63414e66c4d2b0e5544dd5d5f7fba":{"kind":"REWORK","_number":3,"created":"2020-12-15 10:04:15.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/07/767107/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/767107/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/767107/3"}}},"commit":{"parents":[{"commit":"14de07ed463fa9c9f21855b87c810ae3788e7c78","subject":"Remove experimental job legacy-tempest-dsvm-zeromq-multibackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/14de07ed463fa9c9f21855b87c810ae3788e7c78"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 10:03:40.000000000","tz":120},"subject":"Cap bandit to v1.6.2","message":"Cap bandit to v1.6.2\n\nWe don\u0027t have test requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7.\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit 270065e0d42f6c07e9d3b1e875950e4439f7c228)\n(cherry picked from commit 312e81177db4580f388f61574407ce62af366b7f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed6ee8b26ea63414e66c4d2b0e5544dd5d5f7fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed6ee8b26ea63414e66c4d2b0e5544dd5d5f7fba"}]},"branch":"refs/heads/stable/queens"},"7abe6dc063eac6f40a3c4eacde41f222989c7a36":{"kind":"REWORK","_number":4,"created":"2020-12-15 10:35:41.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/07/767107/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/767107/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/767107/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/767107/4"}}},"commit":{"parents":[{"commit":"14de07ed463fa9c9f21855b87c810ae3788e7c78","subject":"Remove experimental job legacy-tempest-dsvm-zeromq-multibackend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/14de07ed463fa9c9f21855b87c810ae3788e7c78"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 09:29:38.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-15 10:35:20.000000000","tz":120},"subject":"Cap bandit to v1.6.2","message":"Cap bandit to v1.6.2\n\nWe don\u0027t have test requirements in upper-constraints, so we need\nto cap it in stable branches manually to work with Python 2.7.\n\nChange-Id: Ie597e778e3efa61bfd98eaaa92730b5050195e7a\n(cherry picked from commit a512bfc233fc76c4375388b8546ab9e60459e1d3)\n(cherry picked from commit 7f77de0519d8d63cea47fcf84831cde625380411)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7abe6dc063eac6f40a3c4eacde41f222989c7a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7abe6dc063eac6f40a3c4eacde41f222989c7a36"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
