)]}'
{"id":"openstack%2Fcinder~767172","triplet_id":"openstack%2Fcinder~master~I667213a1abb025010dc306e4a012dd87c4818756","project":"openstack/cinder","branch":"master","topic":"expand-rep-vol","hashtags":[],"change_id":"I667213a1abb025010dc306e4a012dd87c4818756","subject":"PowerMax Driver - Extend replicated volume","status":"MERGED","created":"2020-12-15 14:56:04.000000000","updated":"2021-03-06 02:47:07.000000000","submitted":"2021-03-05 00:02:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"767172-expand-rep-vol","meta_rev_id":"c2eef2b5852e05869e0d8293e5f8cce313289a3e","_number":767172,"virtual_id_number":767172,"owner":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 00:02:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-04 14:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-06 02:47:07.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":1,"date":"2021-03-06 02:47:07.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2021-03-06 02:47:07.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-06 02:47:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-04 14:28:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19004,"name":"zhimou.peng","email":"zhimou.peng@emc.com"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 16:15:18.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-15 16:15:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 16:24:56.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-15 16:31:16.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 18:20:12.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 19:25:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-15 20:54:57.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 22:22:15.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 00:50:52.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 01:51:14.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 01:54:08.000000000","updated_by":{"_account_id":19004,"name":"zhimou.peng","email":"zhimou.peng@emc.com"},"reviewer":{"_account_id":19004,"name":"zhimou.peng","email":"zhimou.peng@emc.com"},"state":"CC"},{"updated":"2020-12-16 02:03:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 03:18:20.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2020-12-16 03:55:31.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2020-12-16 04:07:58.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2020-12-16 04:16:25.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 04:20:58.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2020-12-16 04:22:09.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2020-12-16 04:22:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 04:29:23.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2020-12-16 05:12:50.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 06:56:17.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2020-12-16 07:06:18.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 17:44:39.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-01-27 18:39:49.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-02-10 12:08:38.000000000","updated_by":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"reviewer":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"state":"REVIEWER"},{"updated":"2021-02-19 14:36:37.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2021-02-19 15:30:04.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2021-02-23 21:36:41.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-03 09:46:38.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-03 11:01:18.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-04 23:32:02.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"}],"messages":[{"id":"c8b9d91e6fc7da93d176d4f82b7913e758b14729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2020-12-15 14:56:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bc3a9ddef4d6dc4dd51e2f84f78d85691a236d8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-15 16:15:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/72/767172/1/check/dsvm-tempest-cisco-zonemanager-job/f4882ff/ : SUCCESS in 1h 14m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ac15da38248a82a74d8ebdb927ed4c18e6d428c3","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-15 16:15:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/72/767172/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/49d717b/49d717b.html : SUCCESS in 1h 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a495a024074c159e8845374bc0b4656cc32c696f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-15 16:24:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767172/1/iscsi/ : SUCCESS in 1h 25m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/767172/1/smb/ : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":1},{"id":"232c59aec2b2a94c77a96ade80c908b027c447af","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-15 16:31:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/72/767172/1/ : SUCCESS in 1h 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"74e6be1909be0a9d29a7124b10bc630b3286eca4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-15 18:20:12.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-767172-1/ns5_nfs : SUCCESS in 3h 23m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-767172-1/ns5_iscsi : SUCCESS in 3h 6m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"781316e26ad5e132d26e9426d3b9243fb4cb3777","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-15 19:25:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/72/767172/1/check/DellEMC_Unity_FC/0f37a82/DellEMC_Unity_FC/299 : FAILURE in 3s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/72/767172/1/check/DellEMC_Unity_iSCSI/7e54259/DellEMC_Unity_iSCSI/284 : SUCCESS in 1h 43m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ba77b59ab9f4894e3644f739c42ae33e2fe49505","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-15 20:54:57.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767172/767172/1/check-not-vote/ext-cinder-zuul/13fe4e3 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"08a3ee2597a677f801fe7dbad3d2c94384bb7057","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-15 22:22:15.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/72/767172/1/check/tempest-dsvm-full-focal-py3/9fba8df/ : SUCCESS in 1h 36m 51s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/72/767172/1/check/devstack-plugin-ceph-dsvm-focal-py3/92ea1f1/ : SUCCESS in 2h 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ebf18594572767fa58c1ea992508f07e757539d3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 00:50:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/72/767172/1/check/DellEMC_XtremIO_FC/2eb5134/DellEMC_XtremIO_FC/321 : FAILURE in 1h 42m 55s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/72/767172/1/check/DellEMC_XtremIO_iSCSI/c36da8a/DellEMC_XtremIO_iSCSI/320 : SUCCESS in 2h 26m 32s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"ebee3331cd5b135e56b514005c7b9fff517446da","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 01:51:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/72/767172/1/20201216084530/NEC-ISCSI : SUCCESS in 2h 01m 45s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/72/767172/1/20201216084530/NEC-FC : SUCCESS in 2h 05m 43s","accounts_in_message":[],"_revision_number":1},{"id":"93c1cd5dbd62870c0c66c9d87222122d437c414b","author":{"_account_id":19004,"name":"zhimou.peng","email":"zhimou.peng@emc.com"},"date":"2020-12-16 01:54:08.000000000","message":"Patch Set 1:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":1},{"id":"6cf3efafb65b5dd7aff4a522bea9f3d7c3f25588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 02:03:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2006c1d10e44a0aa2fd4863eb40407f : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d6ef49d8d748b196566acf88a20cbc : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8766540d1f748428b70f0c0f325ebfc : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c0373c506ee45e2b99ffc2810263a8b : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b8a403e9c494d0a95a089df3ebd6517 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb1595e62c4407fba5de8e4f6e45591 : SUCCESS in 10m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/94d210d9f2a34c9ca5631b4fd383d20d : SUCCESS in 48m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/820d76d74c244973a58fae889de38716 : SUCCESS in 1h 13m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/885a4daaeb7a4f34bf447fe3791b0bf1 : SUCCESS in 12m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3119edec02f746408561b542e17b24f9 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/edd1735b31114af3bee72c17a5c4b642 : SUCCESS in 5m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cac686517341441f83b5729f305f0e39 : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7245a37fc339400d844d137b48478c88 : SUCCESS in 39m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bdd9bc50c80e483dba85753c3d4ae0b3 : FAILURE in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a1985160b4c24a1186df91765101ce28 : SUCCESS in 1h 50m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/df976f731b5a473281abe51c418c2b31 : FAILURE in 1h 35m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40ee95918ca345ad8ad292a66ca3c6a5 : SUCCESS in 1h 35m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b3f473c916714e47a34adda4342f851a : SUCCESS in 1h 24m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/46ea462415da467f8e33708bbd630f41 : SUCCESS in 1h 31m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/182deb9884774579a3cb5f92fcc6c0b3 : SUCCESS in 1h 25m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92f4fde3afa040538bfed9c2ccdfcdc2 : SUCCESS in 2h 07m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/661db6e4a6454c70aad03340a71940a2 : SUCCESS in 2h 25m 00s","accounts_in_message":[],"_revision_number":1},{"id":"78c3eaacc3e72a82ba22d3e8317a12d83166b305","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-12-16 03:18:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/72/767172/1/check/DellEMC_PowerStore_iSCSI/563eb0f/DellEMC_PowerStore_iSCSI/184 : FAILURE in 1h 14m 42s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/72/767172/1/check/DellEMC_PowerStore_FC/762c756/DellEMC_PowerStore_FC/180 : SUCCESS in 1h 06m 22s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc471a7f33cd748c9e2f0c10405f26ade1581a6c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-12-16 03:55:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/72/767172/1/upstream-solidfire-check/cinder-solidfire-iSCSI/d934ba9/ : SUCCESS in 1h 50m 55s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"5ab4a373e6f86e90936e635ed0ffb3e78a41c86a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-12-16 04:07:58.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767172-1/Seagate-iSCSI : SUCCESS in 49:10.00\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767172-1/Seagate-FC : SUCCESS in 41:07.49","accounts_in_message":[],"_revision_number":1},{"id":"0596a5ad3dbb8a44aac2446ba76c672a8bbb511a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-16 04:16:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/72/767172/1/upstream-check/cinder-cDOT-iSCSI/b58672e/ : SUCCESS in 2h 12m 11s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/72/767172/1/upstream-check/cinder-cDOT-iSCSI-IPv6/c50c51d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/72/767172/1/upstream-check/cinder-cDOT-NFS/edfcd84/ : SUCCESS in 37m 42s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/72/767172/1/upstream-check/cinder-cDOT-NFS-IPv6/31f0b4d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/72/767172/1/upstream-check/cinder-cDOT-FCP/f9ea268/ : SUCCESS in 1h 04m 50s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe10f5a41289430ccc44b626e0daae53225cab00","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2020-12-16 04:20:58.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/767172-1/PVME-iSCSI : SUCCESS in 53:03.66\n* PVME-FC http://os-logs.tristero.net/dell-ci/767172-1/PVME-FC : SUCCESS in 47:46.94","accounts_in_message":[],"_revision_number":1},{"id":"b77544cce77487dedc6e4225d077359adeff95ce","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-12-16 04:22:09.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/767172-1/HPMSA-iSCSI : FAILURE in 1:03:31\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/767172-1/HPMSA-FC : SUCCESS in 40:17.33","accounts_in_message":[],"_revision_number":1},{"id":"30d9058ae54a4dc858b62b89960d6a0e1be56461","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 04:22:12.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/72/767172/1/check/DellEMC_PowerFlex/8aeabb2/DellEMC_PowerFlex/144 : SUCCESS in 2h 18m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"d813620ffee1ff8c992f875c6c67ca5f0d196126","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-12-16 04:29:23.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/72/767172/1/check/DellEMC_SC_FC/ebe1e16/DellEMC_SC_FC/132 : SUCCESS in 2h 25m 48s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/72/767172/1/check/DellEMC_SC_iSCSI/cbdc10f/DellEMC_SC_iSCSI/130 : SUCCESS in 1h 34m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"6c16915bead5d9e4a40287b8f82bb7fdae060fbc","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 05:12:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/72/767172/1/check/DellEMC_VNX_iSCSI/ae10b79/DellEMC_VNX_iSCSI/188 : SUCCESS in 3h 09m 15s\n- DellEMC_VNX_FC http://publiclogs.emc.com/72/767172/1/check/DellEMC_VNX_FC/e30d491/DellEMC_VNX_FC/189 : FAILURE in 1h 12m 32s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"87d41a63960dfc35ffb26a8d25cca0f71feb3beb","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-12-16 06:56:17.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/767172-1/Lenovo-iSCSI : SUCCESS in 1:48:33\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/767172-1/Lenovo-FC : SUCCESS in 2:29:17","accounts_in_message":[],"_revision_number":1},{"id":"8969858768e8602684a9e90f3b5b24a269f6b757","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 07:06:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/72/767172/1/check/DellEMC_PowerMAX_FC/7596504/DellEMC_PowerMAX_FC/263 : SUCCESS in 4h 11m 19s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/72/767172/1/check/DellEMC_PowerMAX_iSCSI/2d81e18/DellEMC_PowerMAX_iSCSI/240 : SUCCESS in 5h 08m 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"c80336812b83588e50480faf3dd39c6b14941fd3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-16 08:17:15.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/72/767172/1/check/DellEMC_Unity_FC/135f4a3/DellEMC_Unity_FC/307 : SUCCESS in 1h 27m 19s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/72/767172/1/check/DellEMC_Unity_iSCSI/a288a87/DellEMC_Unity_iSCSI/289 : SUCCESS in 1h 47m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"680b2de9594b3da5cd4b6a47f876c2fcefa15157","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 09:24:56.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/72/767172/1/check/DellEMC_XtremIO_FC/0c68d3f/DellEMC_XtremIO_FC/325 : SUCCESS in 1h 43m 33s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/72/767172/1/check/DellEMC_XtremIO_iSCSI/3827c46/DellEMC_XtremIO_iSCSI/324 : SUCCESS in 2h 32m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"8183215d6b9f25e9bde355f9137183ff89c98258","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-01-25 17:44:39.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"7e078e654c01dad2013e49e80e8ec077b2eaa340","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-01-27 18:39:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b3c480061d07661e434829ae2625e9b92277c7fd","author":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"date":"2021-02-10 12:08:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"410313e2dd6ad702668bb9246af048e88d22a24c","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-02-19 14:36:37.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks Helen, Code looks good to me,","accounts_in_message":[],"_revision_number":1},{"id":"9337f56cf57629a7cdc6bd278b0e5b8f95363578","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2021-02-19 14:40:35.000000000","message":"Patch Set 1:\n\nJust checking, Does this requires a update in drivers version history?","accounts_in_message":[],"_revision_number":1},{"id":"c6c02549abe3539dbb51735e58a957e98ba3679f","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-02-19 15:30:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"369652b8a304a7b2a10e8411ea2f6d8c95f9cf9f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-23 21:36:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI think Fernando has a good point, though this is a weird one because the change isn\u0027t actually in the driver code but is in the utility code.  I think the key question is how will your support people know when an operator is using a driver with this fix vs. one without it?  Suggestion inline, though feel free to reject it (a -1 won\u0027t block a merge, it\u0027s there to get your attention).","accounts_in_message":[],"_revision_number":1},{"id":"99c6497103a577aa8f8582c43c8246a227288996","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-24 22:29:06.000000000","message":"Patch Set 1:\n\nI was just reviewing a different driver and realized that what you should do here is a major version increment to indicate a backward incompatibility (an operator can\u0027t just replace their powermax/iscsi.py with the one on this patch and expect things to work).","accounts_in_message":[],"_revision_number":1},{"id":"cdadc9b4e9e476c6050c192dae25b0685912dd99","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-02-25 10:36:37.000000000","message":"Patch Set 1:\n\nThanks Brian, Fernando,\nI actually did put the increment in the following review which has not merged yet.  Will this suffice?\nhttps://review.opendev.org/c/openstack/cinder/+/761643","accounts_in_message":[],"_revision_number":1},{"id":"213232a3a323139c441e0a74f02394b7b4225d95","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-02 13:30:43.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Thanks Brian, Fernando,\n\u003e I actually did put the increment in the following review which has not merged yet.  Will this suffice?\n\u003e https://review.opendev.org/c/openstack/cinder/+/761643\n\nThis is a big enough change that it should trigger its own revision number, so make this 4.5.0.  It means the patches will have to merge in order, so you may want to put a \u0027depends-on\u0027 in the commit message in case this one is approved first.","accounts_in_message":[],"_revision_number":1},{"id":"1a64213bc64adb417056d9ed3c6b72d4c2c83751","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-02 13:31:23.000000000","message":"Patch Set 1:\n\nAlso, please add a release note.","accounts_in_message":[],"_revision_number":1},{"id":"96360b85e6261a7ac33019a3bf1a955cc26a7c87","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-02 15:15:00.000000000","message":"Patch Set 1: -Code-Review\n\nDropping my -1, this is unreleased code, so the version numbering doesn\u0027t really enter into it.  Release note can be a follow-up.","accounts_in_message":[],"_revision_number":1},{"id":"d5f8c09360615d9f87ac728caeb8c13566fac4d7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-03 01:56:56.000000000","message":"Patch Set 1:\n\nShould probably rebase this on master to pick up the changes from https://review.opendev.org/c/openstack/cinder/+/761643","accounts_in_message":[],"_revision_number":1},{"id":"a194fd4919a245271a2239e401310703f5e791ee","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-03-03 08:53:42.000000000","message":"Patch Set 1:\n\nI will rebase.  Thanks Brian","accounts_in_message":[],"_revision_number":1},{"id":"1872d7bb9eed6b750177ebe3c478efe21c95d815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-03-03 09:16:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"43d83a196e8241801b0ca719df717ac096e19d75","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-03 09:29:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/72/767172/2/check/iscsi-driver-nimble/7cd14bd/7cd14bd.html : FAILURE in 7m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9a31844d68668cceb69bb5ef01c59f2572ad6229","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-03 09:46:38.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/72/767172/2/check/dsvm-tempest-TOYOU-ACS5000-driver/f3ddcc5 : FAILURE in 26m 29s","accounts_in_message":[],"_revision_number":2},{"id":"25f1a98eb5971ed7236c690ca8b86e368cdad31e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 09:56:03.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767172/767172/2/check-not-vote/ext-cinder-zuul/803eacd : SUCCESS in 36m 44s","accounts_in_message":[],"_revision_number":2},{"id":"54beede4b56698efd28f2544e902b8083af50129","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-03 10:38:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/72/767172/2/ : FAILURE in 1h 17m 25s","accounts_in_message":[],"_revision_number":2},{"id":"25dc3b06ce7ee1adae8ecec67e1bf776478fcce9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-03 10:38:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/72/767172/2/check/dsvm-tempest-cisco-zonemanager-job/f4daab9/ : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a1f137af1083b5a4e8492ee99216ae348ce11377","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-03 10:41:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/72/767172/2/check/iscsi-driver-primera/a35dc55/a35dc55.html : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f34cbd703ed5d49b40adb0992654379467095582","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-03 11:01:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/72/767172/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/32b432c/ : FAILURE in 1h 42m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/72/767172/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5f4d5d1/ : SUCCESS in 1h 36m 21s","accounts_in_message":[],"_revision_number":2},{"id":"72adffa8ec07c24604723d595885b3edd787c18c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 11:33:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aa2fb163111483483e2601995728ff3 : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/685134a92d0f467ea187f029024c83d0 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/042c43b373b345cbbbcf3c70598d722b : SUCCESS in 12m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62f3a3a8c94c41388ad17444fd03d893 : SUCCESS in 16m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c17a48920f8346f0a1fb8eb0557efdb5 : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc56a27347845309ba861fb9256c02e : SUCCESS in 17m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/6b3a4fef2a234e84ba3a5284e817d3ce : SUCCESS in 1h 03m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d37267d3292d4ed59999e883d12c15b4 : SUCCESS in 1h 24m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7a7b19731f1e4b1eae77ee5162653e9e : SUCCESS in 13m 05s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6882c4a38a0e4654988e030602571fcf : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/360787ed25c44483955a9d970d17e3a2 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2ceda0d46ff45bc8b771888ce06f74d : SUCCESS in 9m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0812782323a94c388213d65517b3fa35 : SUCCESS in 46m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb730ff82a5846f3b4bf3ed2c7e59470 : SUCCESS in 7m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b96c0ff4d9f47b48ffdae10635e2a89 : SUCCESS in 1h 22m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dff33613e407449f923848be0f7ea148 : FAILURE in 1h 29m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/24ea675a61924b51b225c4b3c291202f : FAILURE in 1h 31m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e472bd3913d34f7991448f1ce7421206 : SUCCESS in 1h 10m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c473789df8c49b49e772aa9c146fc0e : SUCCESS in 1h 31m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f87fe73443144dadb4a7cf5c344d9bb6 : SUCCESS in 1h 20m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b77c3e4933b4b138ad2c2013271ff01 : SUCCESS in 2h 07m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f8d9380e15f41e8b198c973e8f1d4b4 : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":2},{"id":"db980e73a00494e08c85f78e778b842d9b4cea2d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-03 12:43:23.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-767172-2/ns5_nfs : SUCCESS in 3h 25m 37s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-767172-2/ns5_iscsi : SUCCESS in 3h 10m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"745023062e1044ab970d48ddaba1e4db7376ee5d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-03 13:02:19.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f43f08d8ff3645ae8cfc5cee38d5ae7c : POST_FAILURE in 2h 02m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/42f40ec1db64405485021e92955c5567 : POST_FAILURE in 2h 00m 20s","accounts_in_message":[],"_revision_number":2},{"id":"317aed78e16cd961cfc7ccf509f0527d41c4cac2","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-03-03 18:31:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"021f2b67c2d1fc30092acc6c21288e6ea4ad0560","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 19:13:36.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767172/767172/2/check-not-vote/ext-cinder-zuul/56b0478 : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5a9991bf68f185b336ad222379ae87652e75076f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-03 22:28:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/72/767172/2/check/dsvm-tempest-cisco-zonemanager-job/e5bef52/ : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e7a455c8f096ea9518d5b0e3bc58471fc1a15da4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-03 23:24:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/72/767172/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7b40965/ : SUCCESS in 2h 25m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/72/767172/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5045a7a/ : FAILURE in 2h 27m 57s","accounts_in_message":[],"_revision_number":2},{"id":"578647a16c3b46c45117f967417f4e8ed58d7223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:04:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10eeda046e4548d794d2e0e43e07d4e0 : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e25649376f5b416aa7da495abe3ffd9c : SUCCESS in 10m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47c5feec31334c1ab73c23ad8ac70ee8 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e11aeca69b3a42bc89147203aab48775 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2aace14d3cd4a99a9eb621f098b9c65 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a87b5ad1b1422b8215721bb041c51f : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/59ad2dbd55644b5183d50b7b48fa17e8 : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f3d64f3c46e4197a8122248aa123312 : SUCCESS in 1h 06m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/91772830a47640bba88d00c9f0f05191 : SUCCESS in 17m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bfde41a9a8ed463abfa9536ec30064fa : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3e63c19aba24fd28cebd00338bd8e5a : SUCCESS in 6m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb898ed24b944ec4b16912caa51e5575 : SUCCESS in 13m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb754e07108240cd8c911bca13a88b09 : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52d94b3272c14225acabe34ebe1007af : SUCCESS in 7m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04d6fff7969e45a7892385c016aadc46 : SUCCESS in 1h 45m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fa1eb228975240dc93157d0cfc8bd564 : FAILURE in 1h 36m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/974d436c85d04e138e95c5b2802a6062 : FAILURE in 1h 57m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ac30f6af6d74bf5875dbc48d4f6b724 : SUCCESS in 1h 20m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ab42b99f751498a919f7971964d6a7f : SUCCESS in 1h 30m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0bb013b0fbb945c2acd9cf10a30ecc42 : SUCCESS in 1h 20m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a223af24ea849e2be725d326b351d7c : SUCCESS in 2h 04m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e9628abe6e6432fb0a2fc3cb0a6215a : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":2},{"id":"5d7dbe087f5180299eeb09a103153008ad9b3aee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-04 03:23:19.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/77a5d1288a8d488f96db5ce4f1f94259 : TIMED_OUT in 3h 05m 04s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/89154448078e4c9a88fa2a1a1a7e8aab : POST_FAILURE in 2h 06m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f961c09a49768788ebb2ed36d18094f0fd35d74e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 03:49:08.000000000","message":"Patch Set 2:\n\nrecheck cinder-tempest-plugin-lvm-lio-barbican failure is in cinder_tempest_plugin.api.volume.test_volume_backup.VolumesBackupsTest","accounts_in_message":[],"_revision_number":2},{"id":"3d1055e790adf20f3ac62f990020353b1e538fef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-04 04:29:16.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767172/767172/2/check-not-vote/ext-cinder-zuul/e17b8f3 : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":2},{"id":"a6bee8a023bf86c5022a437eeee69cfcffd1db3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 06:05:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1698a96209c14d319015ed067e295527 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a53de41128401c8be01ed49aec3594 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adbe33ef2eeb41e782a47accacdaa274 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac06a6f6a92746ff8e13c3042f823f82 : SUCCESS in 18m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c13b660c54747369a0907f9b1448055 : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6981ebba15f3424499345cc9a410f297 : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/8093c3c6b8c444869dd16cbc64c228fb : SUCCESS in 1h 01m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46d8760b99824922bdbf5365c30a7b12 : SUCCESS in 1h 48m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/69c1ebecb7ce4b3cae79ee24077a211a : SUCCESS in 13m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f7e21d68437497cb738d7089ab96b99 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/434b4292741e45ae88e43a5a7f4f6f9a : SUCCESS in 6m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bf6a54b58e6f4443bf3ffd277302bc64 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0dd2efd241de4969befca3264928001d : SUCCESS in 36m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27fc05bb95d145ed8fa72aaa988d74aa : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/323595511e2e4904a8cc6de812883c44 : SUCCESS in 1h 18m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b91ba237daba474fbb39e0fd8ca56184 : FAILURE in 1h 34m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc62b4a46a134860a8182717cee93d36 : SUCCESS in 1h 27m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/878a3d4d06f14ab8908328201a707349 : SUCCESS in 1h 11m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d7a4cf3cfdce4482b345f3266e9fc6ce : SUCCESS in 1h 14m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6dbb170151b3424d9f95bf8c2e40d2a8 : SUCCESS in 1h 27m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a857156e37144853b52f9314ca68af76 : SUCCESS in 2h 11m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e3d6b8d3b7d40b49cdeb4b64c79c322 : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":2},{"id":"765657fc2dd3f54cf5055c0d330afd74642bc3d0","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-04 06:37:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/72/767172/2/20210304131509/NEC-FC : SUCCESS in 2h 18m 01s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/72/767172/2/20210304131509/NEC-ISCSI : SUCCESS in 2h 22m 21s","accounts_in_message":[],"_revision_number":2},{"id":"09d23889907e10b209825da14c5d24f5dd4d2f71","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-04 08:16:26.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/767172-2/PVME-iSCSI : FAILURE in 15:07.66\n* PVME-FC http://os-logs.tristero.net/dell-ci/767172-2/PVME-FC : FAILURE in 15:12.83","accounts_in_message":[],"_revision_number":2},{"id":"3af6324799f217c45dbff296519ad772d71d80ee","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-04 08:45:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-iSCSI/1cc705a/ : SUCCESS in 38m 40s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-iSCSI-IPv6/807b923/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-NFS/6591e91/ : SUCCESS in 37m 48s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-NFS-IPv6/273433b/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-FCP/ac93e50/ : FAILURE in 1h 34m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"fba4c58a547b0b9d88a6619fb32ec3b27b9663a6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 08:49:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/72/767172/2/check/dsvm-tempest-cisco-zonemanager-job/294acbb/ : SUCCESS in 1h 20m 06s","accounts_in_message":[],"_revision_number":2},{"id":"01f3754229a5cf81ea323a5021f88586efb273da","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-04 10:08:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/72/767172/2/upstream-solidfire-check/cinder-solidfire-iSCSI/f96dcf1/ : FAILURE in 1h 47m 10s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"914bf58e971b28b8233ec378aac42a70c1eacfac","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-04 11:32:16.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/72/767172/2/check/DellEMC_PowerStore_iSCSI/530d3b2/DellEMC_PowerStore_iSCSI/504 : SUCCESS in 53m 48s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/72/767172/2/check/DellEMC_PowerStore_FC/879eaff/DellEMC_PowerStore_FC/498 : SUCCESS in 51m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"14bdafda7783c6a37467b3d29e271583f8092ea2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 14:28:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"27e1cd3ea938709c884ab4457eceb30832a20e57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:30:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e020ad85a67a6fc70e09d0318825d157ce3a634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 16:52:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0120708630241eb90ac88254755d24c : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e47ff439144e65ac7be11bf23e341e : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/849b9187243b4468b5922cd3573d89ba : SUCCESS in 11m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/843972d0484e4829b1396f071c312c2b : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/928deb6bc27544da81f71ad75d927363 : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/6512dee067fb4ca789392f5ce24214c8 : SUCCESS in 1h 07m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff51f9574ad24ef9a6d2d9cbdb27b416 : FAILURE in 1h 15m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2daab43c4f7c4e36b73b19d1fb67ccbc : SUCCESS in 1h 21m 03s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7b304e27eb5c454287955ada4ea31942 : SUCCESS in 1h 27m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b09be37d1ed944849582a96544070477 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1ded1434bd506cb34bf5827790e6fe653e34160d","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-04 17:43:57.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767172-2/Seagate-iSCSI : SUCCESS in 55:07.88\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767172-2/Seagate-FC : SUCCESS in 55:02.85","accounts_in_message":[],"_revision_number":2},{"id":"4b06429fc18bd2df1d78c12d1e0153f86059f495","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-03-04 17:46:44.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/767172-2/Lenovo-iSCSI : SUCCESS in 49:12.43\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/767172-2/Lenovo-FC : SUCCESS in 47:50.95","accounts_in_message":[],"_revision_number":2},{"id":"40bda47ad067f763fb1a9fdf9a1d50a1ad73f820","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 20:23:12.000000000","message":"Patch Set 2:\n\nrecheck tempest-integrated-storage\n\n3 failures:\ntempest.api.volume.test_volumes_clone.VolumesCloneTest.test_create_from_bootable_volume - volume 8c185779-db3c-4a85-b27b-16bb977caad2 failed to build and is in ERROR status\n\ntempest.api.volume.test_volumes_clone.VolumesCloneTest.test_create_from_volume - volume 0b5bad61-b25f-401f-beb5-46757cea3fa3 failed to build and is in ERROR status\n\ntempest.api.volume.admin.test_groups.GroupsV314Test.test_create_group_from_group - volume b4c5a28d-ce11-4e06-9ee6-77d2c2684f87 failed to build and is in ERROR status","accounts_in_message":[],"_revision_number":2},{"id":"7c61656e01b340285e0610d99041d81203c8f517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 22:27:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34888a4204f74f8eb9a1889ddada46f2 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcac9c5608db44f29ecfa18c36ff8fe6 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de40ad9a59f4db39a94a6f0b3d328ef : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6d9ce7777034369b243d8176cea6de3 : SUCCESS in 11m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49365bc6131845cfb5a03a975d721a06 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a5e2bfbe9534a7caa491ec329ba0861 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/78c412888c79485d8e29d0ee109f5dae : SUCCESS in 1h 44m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd221927a39e41dfaaebf4a291530b31 : SUCCESS in 1h 10m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8bb7199e7a9b47648367cb833f5039b1 : SUCCESS in 18m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a940e449a0d1496d833b300c537765c0 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d64b78c4e3d4bb39935d4a8c3dc23b7 : SUCCESS in 7m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2b8afe890c3f410a8f42b0bf6a06c6df : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e0df8281bce44ae9957873273dca0afb : SUCCESS in 41m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ec055651c5e4195a7db91c6aefcd668 : SUCCESS in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3c9df3854084643bfc3eb5dc64de5de : SUCCESS in 1h 40m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b7890ab3adc841f2b9ec5db5854ed649 : FAILURE in 1h 29m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2d162e9bf79841b5934539812b0650f1 : SUCCESS in 1h 44m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8e07a7e3370d4144bab2825b3111b5aa : SUCCESS in 1h 34m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5c33e8edb2214e9d8668fcaf80d763e7 : SUCCESS in 2h 02m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bfe304bc8fb94c55a83a8f04ac49149f : SUCCESS in 1h 57m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92241dfc76bb489b98564487a40c8a13 : SUCCESS in 1h 58m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eec6406e35874598bdba792db45f8967 : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":2},{"id":"ad417842b4c57c95489ec63bc184542bdc0728f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 22:28:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b4fd825d616b53ce0567f0467b5cb3507ab6a40","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 22:32:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/72/767172/2/check/dsvm-tempest-cisco-zonemanager-job/e542c50/ : SUCCESS in 1h 20m 37s","accounts_in_message":[],"_revision_number":2},{"id":"3652c0b52bd8590c825b4bf3bc488bdb4dfaf38d","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-03-04 23:32:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-72-767172-2/1614896872 : SUCCESS in 23m 09s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"ad2539a0b205466d5402050f69398e8e0bf3f817","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-04 23:34:16.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f36b0d8358cc495cbb137eb093cf092a : POST_FAILURE in 2h 10m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/d921438bb4f342958a28ffb5efbb178c : RETRY_LIMIT in 30m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9002b089d502c11d3f71ccf365d261ef930f7e48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 00:02:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b6ac83bd40b47d3a0d3b41f18cabec9 : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7b9a2316ae34e699ac9340fcec0aaad : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f879d3b11fb344ed9936d24675e0c7e7 : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f2652d410b3453db1c075255066a9e9 : SUCCESS in 14m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d885802177764cd2be1b9e73cf317733 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/f0109a73ffb6422187ef67b96cdae6d3 : SUCCESS in 1h 14m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd1da00731ac46059fef15092757db67 : SUCCESS in 1h 18m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2388324cdb324572a5e2d7f9caa6abeb : SUCCESS in 1h 11m 59s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b73434640843460c891bc661ba5da9a8 : SUCCESS in 1h 30m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af33f68ad9ce43ecbe23528473055e25 : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0fb543887cd50edca2da7f9aaa7bc6bf2804e7a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 00:02:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"06cb83230ee047bcddee0d1fcb0d4e4cb5ba3106","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 00:04:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69ebd50dbd24841886f00d364aa6b9c : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2},{"id":"103e7491e8a9cd36cd08a0b709f56425a5186108","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-05 00:38:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-iSCSI/dfbdd22/ : SUCCESS in 43m 26s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-iSCSI-IPv6/8695385/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-NFS/17c856c/ : SUCCESS in 42m 54s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-NFS-IPv6/03db751/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/72/767172/2/upstream-check/cinder-cDOT-FCP/26f4683/ : FAILURE in 2h 09m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b54f3bcff94b769a266804322370681bfe20660f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-05 00:40:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/72/767172/2/upstream-solidfire-check/cinder-solidfire-iSCSI/35fdf26/ : FAILURE in 2h 11m 55s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"8b42d79537f81fd9dc1bb935e58100ddf957ab73","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-05 01:00:40.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/72/767172/2/check/DellEMC_PowerStore_iSCSI/377ed7f/DellEMC_PowerStore_iSCSI/513 : SUCCESS in 58m 55s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/72/767172/2/check/DellEMC_PowerStore_FC/1e47293/DellEMC_PowerStore_FC/507 : SUCCESS in 57m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"3fc9fef101a1560e054aea734974bacf39822319","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-05 01:09:54.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/767172-2/PVME-iSCSI : FAILURE in 15:14.52\n* PVME-FC http://os-logs.tristero.net/dell-ci/767172-2/PVME-FC : FAILURE in 15:20.39","accounts_in_message":[],"_revision_number":2},{"id":"082013cde20b2b36dc816b3edbfac4ed12d27b6a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-05 05:29:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/72/767172/2/check/DellEMC_PowerMAX_FC/9bade79/DellEMC_PowerMAX_FC/528 : SUCCESS in 4h 48m 45s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/72/767172/2/check/DellEMC_PowerMAX_iSCSI/40e3c6a/DellEMC_PowerMAX_iSCSI/493 : SUCCESS in 4h 14m 35s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ab1c695b554a2fe115a5dfed2a9a7f79c4956fa3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-05 06:36:20.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767172/767172/2/check-not-vote/ext-cinder-zuul/8efbe56 : SUCCESS in 36m 55s","accounts_in_message":[],"_revision_number":2},{"id":"27ef742fca85429c71d6c7024035436e5444e928","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-05 06:40:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/72/767172/2/check/DellEMC_PowerFlex/121bd54/DellEMC_PowerFlex/478 : SUCCESS in 2h 39m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"7a9af197870804cf5e045452e7f0620edbdda4a9","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-03-05 12:33:52.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767172-2/Seagate-iSCSI : SUCCESS in 46:05.77\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767172-2/Seagate-FC : SUCCESS in 44:07.86","accounts_in_message":[],"_revision_number":2},{"id":"d85ea202b32077576b468ed8a4058df83c494954","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-06 02:47:07.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/72/767172/2/check/DellEMC_VNX_iSCSI/6939dd4/DellEMC_VNX_iSCSI/486 : SUCCESS in 2h 54m 41s\n- DellEMC_VNX_FC http://publiclogs.emc.com/72/767172/2/check/DellEMC_VNX_FC/fa482d0/DellEMC_VNX_FC/506 : SUCCESS in 1h 26m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"881f72569ac0e509a459f808c18eed06dcf01f10","revisions":{"1f8eade049d8bc17eb885d27b5baf7610270819f":{"kind":"REWORK","_number":1,"created":"2020-12-15 14:56:04.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/72/767172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/767172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/767172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/767172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/767172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/767172/1"}}},"commit":{"parents":[{"commit":"d5750ea093c1931d35933c2f8be898aab74d33b2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5750ea093c1931d35933c2f8be898aab74d33b2"}]}],"author":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2020-12-15 14:55:34.000000000","tz":0},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2020-12-15 14:55:34.000000000","tz":0},"subject":"PowerMax Driver - Extend replicated volume","message":"PowerMax Driver - Extend replicated volume\n\nIn a replicated scenario it is possible for an R2 volume to be larger\nthan its paired R1 if we do not force a new volume creation on both\nsides of the pair.  While using existing volumes is more efficient it\nthrows an exception if we want to extend the R1 to a size that is less\nthan or equal to the R2.\n\nTo fix this we are setting the create_new_volumes flag to true for\nall replicated volumes.  It will remain false for non replicated volumes.\n\nChange-Id: I667213a1abb025010dc306e4a012dd87c4818756\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f8eade049d8bc17eb885d27b5baf7610270819f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f8eade049d8bc17eb885d27b5baf7610270819f"}]},"branch":"refs/heads/master"},"881f72569ac0e509a459f808c18eed06dcf01f10":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-03 09:16:42.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/72/767172/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/767172/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/767172/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/767172/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/767172/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/767172/2"}}},"commit":{"parents":[{"commit":"babafa435c95720a07470e520f6486cd9cdd5acd","subject":"Merge \"Fix CI_WIKI_NAME entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/babafa435c95720a07470e520f6486cd9cdd5acd"}]}],"author":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2020-12-15 14:55:34.000000000","tz":0},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2021-03-03 09:16:37.000000000","tz":0},"subject":"PowerMax Driver - Extend replicated volume","message":"PowerMax Driver - Extend replicated volume\n\nIn a replicated scenario it is possible for an R2 volume to be larger\nthan its paired R1 if we do not force a new volume creation on both\nsides of the pair.  While using existing volumes is more efficient it\nthrows an exception if we want to extend the R1 to a size that is less\nthan or equal to the R2.\n\nTo fix this we are setting the create_new_volumes flag to true for\nall replicated volumes.  It will remain false for non replicated volumes.\n\nChange-Id: I667213a1abb025010dc306e4a012dd87c4818756\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/881f72569ac0e509a459f808c18eed06dcf01f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/881f72569ac0e509a459f808c18eed06dcf01f10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
