)]}'
{"id":"openstack%2Fcinder~767290","triplet_id":"openstack%2Fcinder~master~Ia7d2056e0a49032654812da74dac514dfac83529","project":"openstack/cinder","branch":"master","topic":"bp/toyou-acs5000-driver","hashtags":[],"change_id":"Ia7d2056e0a49032654812da74dac514dfac83529","subject":"Add Cinder driver for TOYOU ACS5000","status":"MERGED","created":"2020-12-16 09:48:03.000000000","updated":"2021-01-22 20:48:40.000000000","submitted":"2021-01-22 20:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":96,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767290-bp/toyou-acs5000-driver","meta_rev_id":"2cc840fa18490b8a63c56882ecec71f65149fbd3","_number":767290,"virtual_id_number":767290,"owner":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 20:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-20 20:21:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-01-21 22:28:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-22 17:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-22 17:04:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32925,"name":"lisa","email":"lisa.qiao@dell.com"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-16 11:10:15.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 11:10:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-16 11:26:05.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 11:28:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-16 11:38:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 13:02:25.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 13:41:16.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 14:33:19.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 16:25:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 16:29:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 21:36:20.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 23:04:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-19 15:47:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-20 11:18:26.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-20 13:51:10.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-21 05:33:24.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-21 05:39:52.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2020-12-21 05:40:31.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2020-12-21 07:27:59.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2020-12-21 07:45:18.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2020-12-21 07:46:42.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2020-12-21 07:52:03.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2020-12-21 07:53:15.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2020-12-21 07:55:45.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-21 09:35:58.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-22 09:19:53.000000000","updated_by":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"reviewer":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"state":"CC"},{"updated":"2021-01-05 16:51:31.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-01-06 14:21:36.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-01-11 02:30:36.000000000","updated_by":{"_account_id":32925,"name":"lisa","email":"lisa.qiao@dell.com"},"reviewer":{"_account_id":32925,"name":"lisa","email":"lisa.qiao@dell.com"},"state":"CC"},{"updated":"2021-01-15 20:41:56.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-01-18 23:55:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"68beb4f9a059aa6294b5d18a8b2996865265d74b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-16 09:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"848e888c22f47a46f0d707b429dd049eecfa3efe","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-16 11:10:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ed0ac88/ed0ac88.html : SUCCESS in 1h 20m 50s","accounts_in_message":[],"_revision_number":1},{"id":"65b1d7bc78aa830f8bfdb7a49aa7b69601786320","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-16 11:10:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/1/check/dsvm-tempest-cisco-zonemanager-job/407aaa7/ : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":1},{"id":"fa907d9183e6c814bd8256257f71c3b9635d1cc5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-16 11:26:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/767290/1/ : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3a2bcef0d3e05883511dd35f42951603b5569ff0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-16 11:28:50.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/1/check/DellEMC_Unity_FC/d8546c1/DellEMC_Unity_FC/309 : SUCCESS in 1h 19m 36s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/1/check/DellEMC_Unity_iSCSI/02cab78/DellEMC_Unity_iSCSI/292 : SUCCESS in 1h 40m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"59ef15fde797a8ae88a769744091ab9c316c09fe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-16 11:38:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/1/check/tempest-dsvm-full-focal-py3/b17e364/ : SUCCESS in 1h 45m 29s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/1/check/devstack-plugin-ceph-dsvm-focal-py3/2c4a238/ : SUCCESS in 1h 43m 49s","accounts_in_message":[],"_revision_number":1},{"id":"307fe63268d3dab449e14e063ac8b8d1917e84b5","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-16 12:14:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/90/767290/1/check/3par-iscsi-driver-primera/a096bf9/a096bf9.html : FAILURE in 2h 25m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8bd0dddce116a1703b02ddb9c1cc3d595f657738","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-16 13:02:25.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-1/ns5_nfs : SUCCESS in 3h 13m 46s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-1/ns5_iscsi : SUCCESS in 3h 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"59e14a72576e22719b9f8fe265aa75bf08b6181c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-16 13:41:16.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"dacdb4ab99ac8db20b6263b69e9fd83890ddb75e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 14:33:19.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/1/check/DellEMC_XtremIO_FC/b9557f8/DellEMC_XtremIO_FC/327 : SUCCESS in 1h 45m 29s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/1/check/DellEMC_XtremIO_iSCSI/41e600b/DellEMC_XtremIO_iSCSI/325 : SUCCESS in 2h 37m 33s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"4c5a255088ef6d685ac7f91204177f1c7ef92e0a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 16:25:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/1/20201216230850/NEC-ISCSI : SUCCESS in 2h 10m 17s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/1/20201216230850/NEC-FC : SUCCESS in 2h 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"77cf610223bfc4ebc2d075682624c77233c75796","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-16 16:29:10.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/1/check-not-vote/ext-cinder-zuul/e3daac0 : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5e08de202aca1f03e00cb79823649f25516f0596","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-16 17:59:01.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9b85af529dee1592bc50a94d84b14dd0b45a045b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-16 21:36:20.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/767290/1/check/DellEMC_PowerMAX_FC/fc38805/DellEMC_PowerMAX_FC/266 : SUCCESS in 5h 00m 51s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/767290/1/check/DellEMC_PowerMAX_iSCSI/624ecc9/DellEMC_PowerMAX_iSCSI/243 : SUCCESS in 4h 34m 08s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"cef6ac1e0cdafa1c466b7491a88ad7ed4b3d2c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 23:04:36.000000000","message":"Patch Set 1: Verified-1\n\n(727 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f32d733cdb74d90b9640d340d9d4153 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c1ab1ffed9b473db40ae3d245a50a00 : FAILURE in 59m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8170e19d619a4abf9b93c8c4a1a99af5 : SUCCESS in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50e41362242046b49c43654f6687fa91 : SUCCESS in 9m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4681c0fad940d2a3971a8cb7a9384d : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa22662bc76b48099a43288c458a8988 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/24e178ed96704f669d551a4da70b418b : SUCCESS in 1h 18m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eef47f823ae34eb59c9801736b88dee0 : SUCCESS in 1h 06m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/92654371fcae4e9097efcc84a0b13497 : SUCCESS in 12m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/39b10423f3914f209aa650fc1f3419ad : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e220b3c9ec114846b4103ff2a8c8185e : SUCCESS in 6m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e7a87256f74c4c88913d7a75d634dbc6 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13103b5375b549b4ab8c9280c9e91eec : SUCCESS in 53m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8845f7dea22c45369b763945e3cd77c2 : FAILURE in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/766a3c7465144199a098c39edd7ed956 : SUCCESS in 1h 15m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c4126e0dd7514f1298b596e7745e9da1 : FAILURE in 1h 13m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6b3424860c048d1a2192ded0852ac50 : SUCCESS in 1h 29m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/87c6201b57854e679746c757c959d91b : SUCCESS in 1h 18m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e66b8538d27242758c19d10bb000ca9f : SUCCESS in 1h 47m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8d4c0105a29341cc8c4ab1cf75e180d0 : SUCCESS in 1h 21m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3717c2fcda99480994e48b830a9b0ee1 : SUCCESS in 2h 13m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aae6c405358d4efd8852050bb46e9d77 : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":1},{"id":"01281dfee598fef52e1429abad0ac30b00ac54b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-16 23:27:27.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c5c07873d70fb627f8785ec04b3cae3a06f3b519","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-19 15:47:31.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027ve seen a few patches pushed. You should stick with this one and update the issues with it.\n\nAt this point, the easiest option is probably to do something like:\n\ngit review -d 767290\ntox -e pep8,py3\n\u003cfix all issues identified\u003e\ngit add .\ngit commit --amend\ngit review","accounts_in_message":[],"_revision_number":1},{"id":"29b947d6a2a5d3686a62ec895073cc3c5b8702d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-20 09:48:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d63f501b3308183ca3b7c2584293cae82ef243b","tag":"autogenerated:gerrit:setPsDescription","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-20 09:51:41.000000000","message":"Description of patch set 2 set to \"format driver code,use command \u0027tox -e pep8\u0027\"","accounts_in_message":[],"_revision_number":2},{"id":"687a203fc28d5b49f6fd04e3df207815ef0b4e61","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-20 10:32:14.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/2/check-not-vote/ext-cinder-zuul/31cee3e : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ae9495c2fcedfa8692973a164ad8ee244cde32d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-20 11:09:10.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e7afe80d36ef5340621a1f4246de439efd163ca5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-20 11:18:26.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/2/iscsi/ : FAILURE in 1h 29m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/2/smb/ : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":2},{"id":"bd73d864846f62ce8a78d11edad282b3b14de0bf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-20 11:33:18.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/2/check/tempest-dsvm-full-focal-py3/a9c5630/ : SUCCESS in 1h 37m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/2/check/devstack-plugin-ceph-dsvm-focal-py3/ee96e9a/ : FAILURE in 26m 04s","accounts_in_message":[],"_revision_number":2},{"id":"98e3f4f7070839b97ba715310e64d65290df3d61","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-20 11:37:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/2/check/DellEMC_Unity_FC/4983673/DellEMC_Unity_FC/347 : SUCCESS in 1h 29m 22s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/2/check/DellEMC_Unity_iSCSI/3b1d872/DellEMC_Unity_iSCSI/324 : SUCCESS in 1h 47m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b5ed3e94d30bed48e253c2e6d585dc088b4386ca","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-20 12:17:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/2/check/dsvm-tempest-cisco-zonemanager-job/b35fe16/ : SUCCESS in 1h 23m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e501ea9a0f12344776ee42ad7bacfcbd8b3a72ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-20 12:34:55.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"bb37e7c0280b7f22960959bf7f13061c9d323a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 12:54:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59f07aeccd4a4cf3970f9f034a5fc832 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e06a606e054b2381fba82504d6df38 : FAILURE in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a83d39334bf4eb3b39de78c1adb3f40 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc13d834dc4743579da59a7be03e2b0b : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f006073b1824756bae8179b401239d9 : FAILURE in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c749265ba14dd18e2c84ff95211c37 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/17e41bd41d2b4e8a8c2de830119bcb34 : SUCCESS in 1h 06m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/633f189bf9d74ec8acde0eca971e12eb : SUCCESS in 1h 06m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6995c0cc6d614fee973be37098570bcd : SUCCESS in 22m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e149ba867d2a461cbc748f78b041ec06 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70e52b8f1b074b4785903ac2244776e8 : SUCCESS in 6m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b908060c7e440e6bdd2962718b0fe0a : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/643cf735197942889ed73d6da6aacaf3 : SUCCESS in 33m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/007fbd252f8c45d388d03e6cf32dca86 : FAILURE in 12m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/09293f2bf27d41fb99f7a86f286f79af : SUCCESS in 1h 02m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7a10c10d7502449d8ead5bf576cfc11b : FAILURE in 1h 11m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/15f313e11b9744d8a262dce5888d33d8 : FAILURE in 1h 17m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e3b59c189c2a4dcfb21a0b09edce621d : SUCCESS in 1h 17m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/971d66ed98f04ff5bc6e0802ff314cb1 : SUCCESS in 1h 02m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a4a6825625a843c389983fc102bad28e : SUCCESS in 1h 27m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0726606ca72a4bd78e1b25fab0d4d08c : TIMED_OUT in 3h 03m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2d4e2ae82554d2cb5d9d22239199bfd : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":2},{"id":"cda3b34bb040c71ef8fa1f37e904fbcfebc4f696","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-20 12:59:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-2/ns5_nfs : SUCCESS in 3h 10m 3s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-2/ns5_iscsi : SUCCESS in 2h 56m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"b9e3c03e440064ace5be72e2631d6c5931d0bc30","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-20 13:51:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/767290/2/check/dsvm-tempest-zadara-driver/39377db : SUCCESS in 1h 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a730e117a31022f8ee633ae9a959802f7ca36b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-20 14:19:22.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4ed4878536aa68ba7adb8de1f8f5d71a502194f3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-20 15:33:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/2/20201220200922/NEC-FC : SUCCESS in 2h 22m 07s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/2/20201220200922/NEC-ISCSI : SUCCESS in 4h 24m 32s","accounts_in_message":[],"_revision_number":2},{"id":"fd78558330857bb89e9924700763ecd485c111c5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-20 21:28:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/767290/2/check/DellEMC_PowerMAX_FC/408a661/DellEMC_PowerMAX_FC/285 : SUCCESS in 4h 22m 49s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/767290/2/check/DellEMC_PowerMAX_iSCSI/9263260/DellEMC_PowerMAX_iSCSI/260 : FAILURE in 5h 13m 43s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"68d379b8b730b5e16edc4dd4a8d0537cd56ba9bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-20 22:56:38.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a2e51a17fdddd2d6e90e8826fe3d4ec5638d1130","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-20 23:07:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/2/check/DellEMC_XtremIO_FC/9e71ed9/DellEMC_XtremIO_FC/358 : SUCCESS in 1h 41m 22s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/2/check/DellEMC_XtremIO_iSCSI/8b4786b/DellEMC_XtremIO_iSCSI/345 : FAILURE in 5h 56m 32s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"de09909bac3334a62946188686dff5df9cb7a6db","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-21 01:36:53.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e I\u0027ve seen a few patches pushed. You should stick with this one and update the issues with it.\n\u003e \n\u003e At this point, the easiest option is probably to do something like:\n\u003e \n\u003e git review -d 767290\n\u003e tox -e pep8,py3\n\u003e \u003cfix all issues identified\u003e\n\u003e git add .\n\u003e git commit --amend\n\u003e git review\n\nThanks,I will try.","accounts_in_message":[],"_revision_number":2},{"id":"fb66fdfe298d793d32ca1f4c810c823dfe8c200e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 01:55:25.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f85dc6fa455929c3bc556b2c11ee5a5e7f0f0ae3","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-21 02:06:38.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e I\u0027ve seen a few patches pushed. You should stick with this one and update the issues with it.\n\u003e \n\u003e At this point, the easiest option is probably to do something like:\n\u003e \n\u003e git review -d 767290\n\u003e tox -e pep8,py3\n\u003e \u003cfix all issues identified\u003e\n\u003e git add .\n\u003e git commit --amend\n\u003e git review\n\nExcuse me.git review -d 767290,then remind me use git pull to merge remote branch into my review\u0027s branch.Do I need to do this?","accounts_in_message":[],"_revision_number":2},{"id":"0dd6ab368c539e09137fbf4bbda73cf034aaa9b7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-21 02:24:16.000000000","message":"Patch Set 2:\n\nIf you need to rebase on master, then you can do:\n\ngit rebase master\n\nOtherwise you shouldn\u0027t really need to do anything other than the \"git review -d\" command.","accounts_in_message":[],"_revision_number":2},{"id":"7ad0c1fa2b51c048d67379d1a513e86c7c063525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-21 02:32:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ba63cc6d0f78ba76a56ff7338babdb9a96de47fb","tag":"autogenerated:gerrit:setPsDescription","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-21 02:36:27.000000000","message":"Description of patch set 3 set to \"use command \u0027tox -e genopts\u0027,add acs5000 driver into cinder/opts.py\"","accounts_in_message":[],"_revision_number":3},{"id":"4a822c67816a3aa51a707e6427eb91585ad5d0a6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-21 03:14:02.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/3/check-not-vote/ext-cinder-zuul/abb980b : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":3},{"id":"2a0ea35e0f4d211090ba4c9982e8053f72e650a4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-21 03:36:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/3/iscsi/ : SUCCESS in 1h 04m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/3/smb/ : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":3},{"id":"16b7302be72d4c3146d7fb36fcd5cd8d3db8e499","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 03:50:02.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b15dbf445375a07fcf17e5d6cceff14fb7abb401","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-21 03:56:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/3/check/dsvm-tempest-cisco-zonemanager-job/12a14f9/ : SUCCESS in 1h 18m 57s","accounts_in_message":[],"_revision_number":3},{"id":"542bb4db1847042e01cb3dfaa1fae2952b7826ba","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-21 04:21:41.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/3/check/tempest-dsvm-full-focal-py3/4e84eae/ : SUCCESS in 1h 42m 58s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/3/check/devstack-plugin-ceph-dsvm-focal-py3/12f8076/ : FAILURE in 28m 04s","accounts_in_message":[],"_revision_number":3},{"id":"92bf35e006f7ed25a5b1c376f088f1ba379ad9ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 04:30:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce1608dc97964b98b94bc8b1cdd66928 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b0ba1e81fa41fa8bd6f1e2cb5fa7eb : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a063204533742869a371beab3b8918a : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23d2238656c74424a8673438a4f38047 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d61781f9af94065bdce2acbf8c8e5d7 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0add2d819820465db8b06a8f77908d5d : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/295d54d5163740968f8e19f73e9b98b3 : SUCCESS in 50m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6248fa34c47c46acac23c08178545afd : SUCCESS in 1h 22m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d477158af4f4a32a8705f36f30b5a58 : SUCCESS in 10m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c9e2b9e9e5fd47b186093ce4b37750ec : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c2906bca6664a54886ef6338093cdf4 : SUCCESS in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/85a229283b5941329ea73571ca0b7406 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fab423bfff9d403b8df1b2e7b7f1c92c : SUCCESS in 33m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0610b4becc224e06b33c7001e4fc1396 : FAILURE in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/96fcf800694c44a1be48162b696973e1 : SUCCESS in 1h 19m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b8117076fc0c400da2817e13b4514196 : FAILURE in 1h 13m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fa641d1e45cf4187b006ef199fc56396 : SUCCESS in 1h 10m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f10e5036462d479d96fbe1dc780db164 : SUCCESS in 1h 06m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/93ad3e7dd813485c8c300680add7e742 : SUCCESS in 1h 35m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e799d717ff6a447a87333c255a8a2417 : SUCCESS in 1h 18m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f61392352134aa995ddcc1e2c90b7b8 : SUCCESS in 1h 54m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2cca9ecd1d64204a0f0519306f3d948 : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":3},{"id":"9611f80463e9b8dae04c2f4280effca5c647677c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-21 04:44:26.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/3/check/DellEMC_Unity_FC/0c0eec0/DellEMC_Unity_FC/350 : SUCCESS in 1h 34m 51s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/3/check/DellEMC_Unity_iSCSI/93c29b3/DellEMC_Unity_iSCSI/327 : SUCCESS in 1h 43m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"67b0cfe5a64635bf759561011ee0963fab67b6a5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-21 05:27:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/3/20201221122354/NEC-ISCSI : SUCCESS in 1h 59m 38s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/3/20201221122354/NEC-FC : SUCCESS in 2h 03m 34s","accounts_in_message":[],"_revision_number":3},{"id":"aea8c8cd1158e70506532efe438a3b285b36762e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-21 05:33:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/3/upstream-check/cinder-cDOT-iSCSI/26bc57f/ : SUCCESS in 36m 39s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/3/upstream-check/cinder-cDOT-iSCSI-IPv6/ab77198/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/3/upstream-check/cinder-cDOT-NFS/a47a642/ : SUCCESS in 38m 23s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/3/upstream-check/cinder-cDOT-NFS-IPv6/af9806a/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/3/upstream-check/cinder-cDOT-FCP/7be828b/ : SUCCESS in 1h 02m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"26afe9cf6455934ade0b654a39ce7ecd26f50f2d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-12-21 05:39:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/3/check/DellEMC_PowerStore_iSCSI/ebccf4b/DellEMC_PowerStore_iSCSI/195 : SUCCESS in 1h 05m 08s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/3/check/DellEMC_PowerStore_FC/1049abe/DellEMC_PowerStore_FC/191 : SUCCESS in 1h 08m 30s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"bcb1d9cae6ac72a3a8fab27db300854446ee1f73","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-12-21 05:40:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/767290/3/upstream-solidfire-check/cinder-solidfire-iSCSI/962487b/ : FAILURE in 1h 08m 57s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"c1e98643c7307b5281009e6a14fa736812f40789","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 05:42:24.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6b485e1407969171d0e5bb921e3d04c8b387b462","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-21 05:54:23.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-3/ns5_nfs : SUCCESS in 3h 21m 34s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-3/ns5_iscsi : SUCCESS in 3h 4m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"36a97126b1b849ff3e980acc4e95c7cb8f2fd4d2","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-21 06:19:05.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e If you need to rebase on master, then you can do:\n\u003e \n\u003e git rebase master\n\u003e \n\u003e Otherwise you shouldn\u0027t really need to do anything other than the \"git review -d\" command.\n\nThanks,I see.\n\nZuul has builded succeeded in patchset 3.It has been verified and plused 1.\nBut openstack-tox-pylint and cinder-plugin-ceph-tempest-mn-aa,these two items are still FAILURE.\nDo I need to deal with?","accounts_in_message":[],"_revision_number":3},{"id":"22661f261e298e3cb14a55611dfd78ef5b41756c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 07:20:26.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"87ab7cda4f5f55c6323e6d03305e10b4fc5d245a","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-12-21 07:27:59.000000000","message":"Patch Set 3:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-3/Seagate-iSCSI : SUCCESS in 48:30.21\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-3/Seagate-FC : SUCCESS in 39:54.82","accounts_in_message":[],"_revision_number":3},{"id":"0c0f38c343f8554625c721aca348caba1da18a49","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2020-12-21 07:45:18.000000000","message":"Patch Set 3:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/767290-3/PVME-iSCSI : SUCCESS in 49:11.21\n* PVME-FC http://os-logs.tristero.net/dell-ci/767290-3/PVME-FC : SUCCESS in 48:01.22","accounts_in_message":[],"_revision_number":3},{"id":"244595fefbc730d18fe6302c56de84b2c8864bf2","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-12-21 07:46:42.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/767290/3/check/DellEMC_SC_FC/41f64de/DellEMC_SC_FC/143 : SUCCESS in 2h 05m 34s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/767290/3/check/DellEMC_SC_iSCSI/0b42843/DellEMC_SC_iSCSI/141 : SUCCESS in 1h 28m 05s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"fed08a07ffb1aa91d7b1b9f0fd5e6b3d1b55b8df","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-12-21 07:52:03.000000000","message":"Patch Set 3:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/767290-3/HPMSA-iSCSI : SUCCESS in 1:01:23\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/767290-3/HPMSA-FC : SUCCESS in 36:12.55","accounts_in_message":[],"_revision_number":3},{"id":"b3cc9551c569c46d5beeba121b03c36d2221bbbe","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-12-21 07:53:15.000000000","message":"Patch Set 3:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/767290-3/Lenovo-iSCSI : SUCCESS in 51:43.75\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/767290-3/Lenovo-FC : SUCCESS in 50:38.47","accounts_in_message":[],"_revision_number":3},{"id":"6aa75d24fd554daa0bacb49021911a396f16fc9b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-12-21 07:55:45.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/3/check/DellEMC_PowerFlex/d53aeae/DellEMC_PowerFlex/155 : SUCCESS in 2h 15m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"adcb55d799c7983264ebe6bb57643acba03deaad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 08:56:00.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2cd943c8b51f8f36732a076bce5582557735cd9d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-21 09:35:58.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/767290/3/check/DellEMC_VNX_iSCSI/5b059ef/DellEMC_VNX_iSCSI/198 : SUCCESS in 3h 02m 05s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/767290/3/check/DellEMC_VNX_FC/9e7fcb7/DellEMC_VNX_FC/199 : SUCCESS in 1h 03m 29s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"b71d0425808885cf69274661ea870fd6632be5ea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 10:47:51.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"9be4cbb890c89957080481148fab0ccebeca4018","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-21 11:42:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/767290/3/check/DellEMC_PowerMAX_FC/16e7fa1/DellEMC_PowerMAX_FC/287 : SUCCESS in 3h 51m 46s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/767290/3/check/DellEMC_PowerMAX_iSCSI/1cf2756/DellEMC_PowerMAX_iSCSI/262 : FAILURE in 5h 23m 36s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"9c4b33cc14a44e6831ec4795f4021345ed03a600","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 13:00:02.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"fa3f7668a420a4f9039ebb77293f09e316933d8c","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-22 09:19:53.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"2d2d4f132cd908fc894cb8e00cb0ddf677a8fcdd","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-22 10:20:08.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"4eceae94442af835ad2f8f79cebf512a6cbfee9c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-22 11:06:51.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"aa034b79e6e3e4c177e6ee37fd88ba3324a7d8c5","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-23 02:14:58.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"13057d62dda7fb4f3390bcb56194b58e5ccc00c1","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-23 03:20:58.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"8a3e62968ab4ceda4a25f5efa2dd6c6e7fcd3738","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-23 03:29:50.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"3c2a633d02a7a86fa0b57f837ad1d3d41102a9b3","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-24 03:24:39.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"49cf5696b45ab6aee6236dcb3a1c89bfec266412","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-24 03:29:42.000000000","message":"Patch Set 3:\n\nOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"77711ab983ad8cab414f9e7ef7f130244c781095","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-24 03:32:28.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"75c10744a06c7c19cac9b0a2ab3f4ec2d8881cc5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-24 04:48:42.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"24a802764a0f5a99ea870fb0ce211dedd004720d","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-24 09:14:44.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"30b5d6086e2d9c5f5921a4a810d869e420f55002","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-24 09:30:29.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"412e26002433d9a841ad654809a7d5ba83fda610","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-24 10:33:52.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"16d2fd2768cc6b745306d2de66eb0ae760e36310","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-25 07:21:42.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"5fe578931200817f7e2e582cacc739b34ec7368f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-25 08:32:26.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ef546fd0086cecfd349fb0d21dd4808d4b7c863e","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-25 09:49:52.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"28ca9461a488f3dd3b5f0bc94db182ef457fffa4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-25 11:08:40.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"94287645aad8fd1029add3a4d72ae4a5c90ff41f","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-28 01:48:17.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"05bf7753a179e660103f2333ef2f8bfcc20ccf8d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-28 05:11:17.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"5c2f4e6f341752960502b7b5f66dbc52bb50b530","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-28 07:39:36.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"fc9a3e95f11bd64f588336da649c5e09eb014d50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-28 10:04:37.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"4ca56b4d0e7596b25c6d181e57331dbabcc374e3","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-28 16:47:44.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"ed981e243e0bd1a215cd069ffe23a390bdabeb42","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-28 18:00:10.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"4951ad32e1cc382bfe4ed4651797ded66fe6150f","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2020-12-29 06:07:12.000000000","message":"Patch Set 3:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":3},{"id":"23e912b117f42d8154bce4e4dae7d897797f36ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-29 08:28:20.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ff786345a4cdc4418d48f65876965e21e9af5b5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2020-12-29 09:07:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"73f393a7c3434600c02500f64d9cdb55df65edb7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-29 09:31:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/767290/4/check/dsvm-tempest-zadara-driver/8860503 : ABORTED in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3d54f2d480d03eae35bc1b78efab0e69b6592b2b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-29 09:48:28.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/4/iscsi/ : FAILURE in 40m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/4/smb/ : FAILURE in 33m 13s","accounts_in_message":[],"_revision_number":4},{"id":"a0ca6d57147dd76ecd53145a9ff826585f900d1d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-29 09:49:14.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"d6438d761acf29a1a25912e3ecb3b946ce7cebfb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-29 09:52:03.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/4/check-not-vote/ext-cinder-zuul/b9a6f69 : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":4},{"id":"7ad6b6cc1fc129d856a4dddc0dec9b239decb5ad","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-29 10:29:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/4/check/dsvm-tempest-cisco-zonemanager-job/b8568b5/ : SUCCESS in 1h 17m 05s","accounts_in_message":[],"_revision_number":4},{"id":"6e4670bf587f07356bf00e99829dec9fad88f37f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-29 10:48:18.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/4/check/tempest-dsvm-full-focal-py3/f25034a/ : SUCCESS in 1h 32m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/4/check/devstack-plugin-ceph-dsvm-focal-py3/02caed1/ : SUCCESS in 1h 34m 36s","accounts_in_message":[],"_revision_number":4},{"id":"c1239ca26225b3e0b49555c5f5b1384823096c47","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-29 10:48:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/767290/4/ : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":4},{"id":"4f369fc3211cafb799d9cb1daccf5cbffb674617","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-29 11:02:57.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"da69fa047e2d59db892a82315478713e9652a6fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 11:10:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09643bd8e98b4a0ca90d024a2c43a982 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f2770d1c2249e3a17c4be6930f155e : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b99e09f2acc7449180eb95bdc5acd313 : SUCCESS in 9m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dddb4803876f4e11ac23abd074e91a76 : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/902be84d901946449ee5ca373f371070 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05eadd4936d46d18ec349847852d7e0 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/8e6b7ea37cdb40c2bc224efef5689776 : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f6bb73f98144c1889b600b338c6e419 : SUCCESS in 1h 11m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e04d28311b82475f8c719f1bc04fcff9 : SUCCESS in 12m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/43daa31ff13d4773ab530713e3c1d2ef : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1eece124d1854c43b6ebb66a11d85510 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d1ac0ae53bd447ee8e517505a240df35 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/605009692616456f8a9115b73847bc7e : SUCCESS in 43m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/076d85ff5c1d45be9452fafb1b7a64d1 : FAILURE in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/217574c4391f433f966351e7c7910699 : SUCCESS in 1h 03m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1caa2c3323224ac3a173b6bd618814b6 : FAILURE in 1h 13m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5210b8eacf5f4b7eb9a6c0ffffadb71a : FAILURE in 1h 15m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/625e1e13140b41b2abfe39c4dcfe5e6b : SUCCESS in 58m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/05092daeab974602941dbf3ac78fb7b0 : FAILURE in 2h 00m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/38996aa667fb47ae9bd5f0c47cbf57b6 : SUCCESS in 54m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5454098fb8c4443a03b455513037dfc : SUCCESS in 1h 55m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5992513ab8244fca8b2ec0d4021ae9a1 : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":4},{"id":"8117ef9fc79bf0b439e4bc7a8bc98a198c56cc3d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-29 11:48:48.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/4/check/DellEMC_XtremIO_FC/ed203e9/DellEMC_XtremIO_FC/398 : SUCCESS in 1h 42m 35s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/4/check/DellEMC_XtremIO_iSCSI/a031e7c/DellEMC_XtremIO_iSCSI/384 : SUCCESS in 1h 38m 08s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"f90b3a564d879d78044c4aa7d1c7c53c90c71cd2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-29 12:23:04.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-4/ns5_nfs : SUCCESS in 3h 15m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-4/ns5_iscsi : SUCCESS in 2h 59m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6d83869edb0f07cc24b38372fb00c4f317eb6bd1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-29 13:04:07.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"a4eb210480eda2c6a6b38601b3ac1162465c6e20","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-29 14:03:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/4/check/DellEMC_Unity_FC/94a8172/DellEMC_Unity_FC/382 : FAILURE in 2h 50m 24s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/4/check/DellEMC_Unity_iSCSI/e47be06/DellEMC_Unity_iSCSI/360 : SUCCESS in 1h 49m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b75b182f2ef6bfbc2a5756f6dfebbb5bcdd8761b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-29 15:18:20.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"7ab4cf8d68e102c617c7b58f128839fd567f3322","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-29 18:17:31.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/767290/4/check/DellEMC_PowerMAX_FC/a20299c/DellEMC_PowerMAX_FC/320 : SUCCESS in 4h 26m 53s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/767290/4/check/DellEMC_PowerMAX_iSCSI/e8ce097/DellEMC_PowerMAX_iSCSI/293 : SUCCESS in 4h 58m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"b3ce52b05fa7cf5a760e494d3ba41aa467ad7bd6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-29 19:31:10.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"1d92c578745856449f0963f133891008b07a23cc","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-04 02:44:00.000000000","message":"Patch Set 4:\n\nrun-ZUUL","accounts_in_message":[],"_revision_number":4},{"id":"11043dc25ab8da4954adc7a6e5c5fdaf70fe4572","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-04 02:44:22.000000000","message":"Patch Set 4:\n\nrun-Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4fb7405e4ba276955e7ea39f5d16a109e3cd558a","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-04 07:19:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ee3c55927e164715fcfda1ceba235bbc790e7394","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-04 07:19:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a0632e445f2ca39b1a9c39806aa722bf9337fd44","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-04 08:01:28.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/4/iscsi/ : FAILURE in 41m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/4/smb/ : FAILURE in 35m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e06f3b41d05b60c2648691ee1ed427875da46eaa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-04 08:02:47.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/4/check-not-vote/ext-cinder-zuul/b444d54 : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":4},{"id":"c973f795835a4a2a12f10bd836ef84e96c2e79b2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-04 08:39:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/4/check/dsvm-tempest-cisco-zonemanager-job/154c258/ : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":4},{"id":"37098aa1f4d61d3cc1c6f7a5b1bc94d11750f8d2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-04 09:01:57.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/4/check/tempest-dsvm-full-focal-py3/bbb90ab/ : SUCCESS in 1h 35m 41s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/4/check/devstack-plugin-ceph-dsvm-focal-py3/e8d296b/ : SUCCESS in 1h 26m 25s","accounts_in_message":[],"_revision_number":4},{"id":"99e497b8218b2f92ee83a400edc96d9d822da918","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-04 09:06:49.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"91a404d3e51b154ea44077a8dcd5055b36540dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 09:12:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1140ff24be14481e8c744a92650d5041 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc3281fa62a2404ea340067ff2a4bc7f : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53bea4aac45d44d68cec54142daaa29b : SUCCESS in 9m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7c4a8c9fac2488eaa327285da38b22d : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3852a0d1f35f4bcea491e00035d09576 : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db02dc1ec05641ed87d7f03d026f05b0 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/c4cfa71786a241a6850ee05114f040a8 : SUCCESS in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1fde31b05eec47ce8fa55aee56ab2b19 : SUCCESS in 57m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6223fb716df44b76a82502c979d3fe73 : SUCCESS in 10m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04e44c83256c432ba22f00467296961b : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0d97a941be04ee184726535cffc58ab : SUCCESS in 5m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/10091345b1bd4acca1d9563b51a3e68f : SUCCESS in 5m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f1df88102a10451590c94c6c07e619d5 : SUCCESS in 44m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e4a8f099915c489691b822c529c6a82a : FAILURE in 4m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15cf47b028d64b6ca9545c3f7403ab3f : SUCCESS in 59m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/269ed0c50d374515a6130368e0b10c52 : FAILURE in 1h 10m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e907135ae5734bb98f3f949b00e3144b : SUCCESS in 1h 29m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3a9e6ac6f64e425e900c405f4aa78fd0 : SUCCESS in 1h 17m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/64a82e179c9b4167a87be3c962a74e9a : SUCCESS in 1h 33m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/21c9aa74799d435281dd17f9d7d10726 : SUCCESS in 1h 34m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b5ce380fcdb41dc8fb72d9e92d609d4 : SUCCESS in 1h 51m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/164efbba474d4796a605d0af3e5c678a : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":4},{"id":"fb7a113295464dcc62ba9dff3631f5943d5a5299","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-04 10:08:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/767290/4/upstream-solidfire-check/cinder-solidfire-iSCSI/dc580f6/ : SUCCESS in 54m 39s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"218848fd53a4e5764045e67c8b2edd115b461f3b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-04 10:33:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/4/upstream-check/cinder-cDOT-iSCSI/b9d93f5/ : SUCCESS in 37m 35s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/4/upstream-check/cinder-cDOT-iSCSI-IPv6/09093a8/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/4/upstream-check/cinder-cDOT-NFS/7d272ae/ : SUCCESS in 39m 16s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/4/upstream-check/cinder-cDOT-NFS-IPv6/016494b/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/4/upstream-check/cinder-cDOT-FCP/63f9695/ : SUCCESS in 1h 19m 38s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"f59400d9f4ff724d3700d2a7d0ef7dc9ca68f376","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-04 10:54:33.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/4/check/DellEMC_PowerStore_iSCSI/00e97b4/DellEMC_PowerStore_iSCSI/222 : SUCCESS in 53m 02s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/4/check/DellEMC_PowerStore_FC/ad48008/DellEMC_PowerStore_FC/216 : SUCCESS in 49m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"05acd86945d1e5836954910b7951fc9a8ff93615","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-04 11:19:40.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-4/Seagate-iSCSI : FAILURE in 49:16.01\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-4/Seagate-FC : SUCCESS in 41:11.80","accounts_in_message":[],"_revision_number":4},{"id":"73e86a549d26c89ee2becbbcb8be5bb009bd9fed","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-01-04 11:27:44.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/767290-4/PVME-iSCSI : SUCCESS in 50:24.82\n* PVME-FC http://os-logs.tristero.net/dell-ci/767290-4/PVME-FC : SUCCESS in 48:55.79","accounts_in_message":[],"_revision_number":4},{"id":"6983176ca60096e088126acb5046ad0c20468531","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-01-04 11:29:33.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/767290-4/HPMSA-iSCSI : FAILURE in 1:05:21\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/767290-4/HPMSA-FC : SUCCESS in 36:52.50","accounts_in_message":[],"_revision_number":4},{"id":"4ad83f84024ace203971b90521c4d70129240539","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 12:19:42.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/4/check/DellEMC_PowerFlex/f77bbd0/DellEMC_PowerFlex/180 : SUCCESS in 2h 15m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"67ccca574adf309bb6ecd7d9dfbbf5ffdc9c82a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-04 13:05:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"718d5ac83a971fd8dfdd8414643f83853339c831","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 13:08:12.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/4/check/DellEMC_XtremIO_FC/0eaae1b/DellEMC_XtremIO_FC/440 : SUCCESS in 1h 40m 46s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/4/check/DellEMC_XtremIO_iSCSI/b156ce7/DellEMC_XtremIO_iSCSI/425 : SUCCESS in 1h 34m 41s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"6ca8a71cc18059f072e231407dbff2f83ee88be9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 13:53:38.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/767290/4/check/DellEMC_VNX_iSCSI/09c33fb/DellEMC_VNX_iSCSI/224 : SUCCESS in 3h 06m 14s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/767290/4/check/DellEMC_VNX_FC/d5bc5aa/DellEMC_VNX_FC/225 : SUCCESS in 1h 08m 26s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"dc8e6e9c0b485fec196faf8237ed220e0756ba79","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-04 14:32:04.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"eee08f4e945cc33a86227e6738f016894116e1a3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 18:48:22.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/767290/4/check/DellEMC_PowerMAX_FC/87afcf4/DellEMC_PowerMAX_FC/360 : SUCCESS in 3h 57m 11s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/767290/4/check/DellEMC_PowerMAX_iSCSI/e6de990/DellEMC_PowerMAX_iSCSI/331 : FAILURE in 4h 22m 46s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"7432a1e5fe211bea5dd3c5a59006e3c4cf9b69e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-04 20:38:52.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"3a2a7c03b16b84bffde5fd47b024d62fae1120e6","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-05 06:34:21.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"41e9fcd8d4426588d4756e577b6de20be75d6c13","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-05 06:36:22.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"73b87c956c75e4b3e943be9cb171e2b8ac28531c","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-05 06:41:07.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"dcb2f68c7743dec0ea5177c62dfc369361ede6b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-05 08:36:43.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"47bc3c3c0a565f75c621ec16fe32d92df8d05f1e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-05 16:51:31.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"657a6617c956a0ca0c1e646e963b6c6623556b77","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-05 19:02:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.  To recheck use \u0027TOYOU ACS5000 CI: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/4/check/dsvm-tempest-TOYOU-ACS5000-driver/8714ec6 : SUCCESS in 1h 46m 22s","accounts_in_message":[],"_revision_number":4},{"id":"b4128ed4c9cc81ee91c0f42a6db97a86d01a20af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-05 19:06:16.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"bce3aaec762d00185eb1815aad660c212d848188","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-06 11:44:50.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"45a7b3ffa07923e84da75e2b5f2ff6c73af32c2a","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-06 12:44:46.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027TOYOU ACS5000 CI: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/4/check/dsvm-tempest-TOYOU-ACS5000-driver/09399e7 : FAILURE in 57m 35s","accounts_in_message":[],"_revision_number":4},{"id":"23c6cf906334f4a667c3dafa8eda8ca7af80a9d9","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-06 12:47:11.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"9a949b1a7a5517a03738ff152ebd21681734fad5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-06 13:22:42.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"1efac93ffa8fb66c6c9b5c6ffba1abf554201d96","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-06 13:34:03.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027TOYOU ACS5000 CI: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/4/check/dsvm-tempest-TOYOU-ACS5000-driver/4e96829 : FAILURE in 39m 40s","accounts_in_message":[],"_revision_number":4},{"id":"9daf885af41e57728384a580f0aac5a0f58851b7","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-06 13:44:39.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"9eef91a83d30e4051dfa08a8f77eba036f2cb955","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-06 13:51:06.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027TOYOU ACS5000 CI: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/4/check/dsvm-tempest-TOYOU-ACS5000-driver/9bb81a8 : FAILURE in 2m 23s","accounts_in_message":[],"_revision_number":4},{"id":"9683cf2576898a5b03e601094c7902ac5ebc716d","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-06 13:54:10.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"b3ea12f549ac8080771448dd2b9d0c37dd8649c7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-06 14:00:07.000000000","message":"Patch Set 4:\n\nMake sure your CI supports the required recheck comment of \"run-TOYOU ACS5000 CI\".\n\nYou should get unit tests added soon. That may actually help uncover issues that will make running CI tests more likely to pass.\n\nDon\u0027t forget the documentation too.\n\nhttps://wiki.openstack.org/wiki/Cinder/how-to-contribute-a-driver\nhttps://docs.openstack.org/cinder/latest/contributor/drivers.html#new-driver-review-checklist","accounts_in_message":[],"_revision_number":4},{"id":"436106289d2bec1aad5913fda8a70b1ec885a337","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-01-06 14:21:36.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d26b343e2acbea17a594d159c9e92e5fb6d44031","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-06 16:56:51.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"80dd00a459afe129597d1d8b6e395c4baa1b1ad6","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-06 17:14:50.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027TOYOU ACS5000 CI: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/4/check/dsvm-tempest-TOYOU-ACS5000-driver/525a359 : FAILURE in 48m 04s","accounts_in_message":[],"_revision_number":4},{"id":"63a2548a6cb0e97eb7c34de9dd2b1eaedc0260ab","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-06 17:39:37.000000000","message":"Patch Set 4:\n\nTOYOU ACS5000 CI: recheck","accounts_in_message":[],"_revision_number":4},{"id":"668f44e5a5363dbe039201cba1b3b4b8d1b253fe","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-06 19:11:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.  To recheck use \u0027TOYOU ACS5000 CI: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/4/check/dsvm-tempest-TOYOU-ACS5000-driver/c8dbd7c : SUCCESS in 1h 21m 39s","accounts_in_message":[],"_revision_number":4},{"id":"10ae3b43039d0be5d9862141df6102913dd5e5fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-07 02:25:08.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"dd37b8a0e1018276daad8d87adfdcad22abc5863","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-07 02:36:40.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e Make sure your CI supports the required recheck comment of \"run-TOYOU ACS5000 CI\".\n\u003e \n\u003e You should get unit tests added soon. That may actually help uncover issues that will make running CI tests more likely to pass.\n\u003e \n\u003e Don\u0027t forget the documentation too.\n\u003e \n\u003e https://wiki.openstack.org/wiki/Cinder/how-to-contribute-a-driver\n\u003e https://docs.openstack.org/cinder/latest/contributor/drivers.html#new-driver-review-checklist\n\nSean McGinnis: Thanks for your review. I will change the recheck comment filter rules, add unit tests and documentation as soon as possible.","accounts_in_message":[],"_revision_number":4},{"id":"68a90094e1207dd0f07b35b8161576ee5cf45ec8","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-07 02:54:11.000000000","message":"Patch Set 4:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":4},{"id":"05c19ad0cf684d031183ed8a81b80d4d168714f6","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-07 04:22:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/4/check/dsvm-tempest-TOYOU-ACS5000-driver/d411cc5 : SUCCESS in 1h 25m 22s","accounts_in_message":[],"_revision_number":4},{"id":"9e6a71d38fdc47b6bc03ded5d841e4ea284ac14a","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-07 06:14:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d6034cc14fb31bc3a6e0e5b88e5ac14870996d88","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-07 06:24:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bcb22956dc6e89b88513a832b9f3a6a74b9e8059","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-07 10:24:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8b462fa236d3018fedca616ec2d7db48fdc69d7e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-07 10:24:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e4af3372722f0e39fb64ce2690463e09f0ed1dc8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-07 10:36:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera finger://openstackcifc-virtual-machine/af36ab6994924ff2be6263e1f97e4547 : RETRY_LIMIT in 5s","accounts_in_message":[],"_revision_number":5},{"id":"1f36e690d104fccb09c4f553e641429447dfc0db","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-07 11:06:11.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/5/check-not-vote/ext-cinder-zuul/77a9874 : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":5},{"id":"0e29f68d10dbcb008ac2e30d3f48a4b5f770a141","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-07 11:20:59.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/bc3d247/bc3d247.html : FAILURE in 55m 19s","accounts_in_message":[],"_revision_number":5},{"id":"611a66b16572ab8e0186d591abc93ea421546633","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-07 11:37:23.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/5/iscsi/ : FAILURE in 1h 12m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/5/smb/ : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":5},{"id":"26d4f6dd12d5d67a4dbad529b2c86e40365f1862","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-07 11:39:27.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/5/check/dsvm-tempest-TOYOU-ACS5000-driver/b023baf : FAILURE in 1h 13m 17s","accounts_in_message":[],"_revision_number":5},{"id":"dea44b3c03787c0f41df9d5c7daeb362384dc702","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-07 11:45:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/5/check/dsvm-tempest-cisco-zonemanager-job/2f612da/ : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":5},{"id":"2ba9aa26f89936350b803e0385fb7dbb6e7c5660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 12:35:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee91bc1581644d1c9f5e4e9616e247cd : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd16c09063b4276a693fcf8f8f85456 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41db57c9f18347ac9de4064304a0e555 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f02cdcf2c75a4d7084e37cbc630c572e : SUCCESS in 9m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/911077e66ae048859b20c8daf9451e32 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24fd3347999d488cbef8a961c7ec22a7 : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/1732d2b880a34239bde4845c3cdadedd : SUCCESS in 57m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf1d151dbdf9474bb8d47a8fdbff2c2d : SUCCESS in 1h 17m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/826a8cefa73843bf90df352a8d35e54e : SUCCESS in 12m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/823766915f504c6e83480805e7127307 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee0bbbe7414344e1aa078d09343a9997 : SUCCESS in 6m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1aafc7e069fb464dbe9be61455836d4f : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9bf566661aed43579ab8f7d60661aa5d : SUCCESS in 48m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff38d8b2e6224e2f91c16db855ec3622 : FAILURE in 4m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed5f90f7842e4cecb959dbce45b9bd5d : SUCCESS in 1h 18m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/46c9e7853a28439d90f2b8a25557c63b : FAILURE in 1h 15m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5ea1151ead6467b90bd02401d78ada5 : SUCCESS in 1h 32m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9cede82282304f3eb9a16cee471df876 : SUCCESS in 1h 06m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ce9d4879726f43cdad6f260fea7fd5d4 : SUCCESS in 1h 25m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/616943a9803949ecaad9b608f71adba9 : SUCCESS in 1h 22m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd282a020b3345f3ab629998fad6cf4c : SUCCESS in 2h 09m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d048984ab9514cb594b0e5199ec5e755 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":5},{"id":"07f728ae0954e92dc296b4aea3d9c28a0a2a7c0c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-07 13:23:44.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/5/check/tempest-dsvm-full-focal-py3/a61bc80/ : SUCCESS in 2h 55m 18s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/5/check/devstack-plugin-ceph-dsvm-focal-py3/0b5b843/ : SUCCESS in 2h 02m 30s","accounts_in_message":[],"_revision_number":5},{"id":"f95a2476354fb6c8fb635aa8a5802bdfb213eb27","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-07 13:29:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/767290/5/upstream-solidfire-check/cinder-solidfire-iSCSI/05f8db1/ : SUCCESS in 53m 00s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"83798e568e356f8e25ec143652c95ff7b3202491","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-07 13:46:36.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-5/ns5_nfs : SUCCESS in 3h 21m 49s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-5/ns5_iscsi : SUCCESS in 2h 50m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"3c2e7afc4632e309ba732d6b2e0e6f2eace5e6f2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-07 14:09:46.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/5/check/DellEMC_PowerStore_iSCSI/31b077f/DellEMC_PowerStore_iSCSI/244 : SUCCESS in 50m 28s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/5/check/DellEMC_PowerStore_FC/1284b87/DellEMC_PowerStore_FC/238 : SUCCESS in 50m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"72c1b94802f3ab7ca407c5391a742d7a157291ec","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-07 14:38:11.000000000","message":"Patch Set 5:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-5/Seagate-iSCSI : SUCCESS in 47:10.93\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-5/Seagate-FC : SUCCESS in 41:13.85","accounts_in_message":[],"_revision_number":5},{"id":"9d917def39b17cda53b20ce648d7db3213702556","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-07 15:10:17.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/5/check/DellEMC_XtremIO_FC/18d3ea9/DellEMC_XtremIO_FC/468 : FAILURE in 2h 05m 13s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/5/check/DellEMC_XtremIO_iSCSI/30814b6/DellEMC_XtremIO_iSCSI/455 : SUCCESS in 1h 29m 18s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"24787cf2be4bd59c57f2e162d6cebd520c34014f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-07 15:40:55.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/5/check/DellEMC_PowerFlex/ceb413c/DellEMC_PowerFlex/201 : SUCCESS in 2h 15m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"b8649a5d24af78653a9dca55a99f91f2fdf1fe71","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-07 15:46:25.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"b799511cf6d8e8ffc3211c4838f20ee3a7242484","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-07 15:52:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/5/upstream-check/cinder-cDOT-iSCSI/e575bf5/ : SUCCESS in 39m 23s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/5/upstream-check/cinder-cDOT-iSCSI-IPv6/b06f5f8/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/5/upstream-check/cinder-cDOT-NFS/26acce2/ : SUCCESS in 37m 06s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/5/upstream-check/cinder-cDOT-NFS-IPv6/698ad12/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/5/upstream-check/cinder-cDOT-FCP/6c83641/ : FAILURE in 3h 16m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"9ca09b11456fc2ddd873a0e750587710f6ba2bea","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-07 15:53:03.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/767290/5/check/DellEMC_SC_FC/e20a283/DellEMC_SC_FC/189 : FAILURE in 2h 12m 11s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/767290/5/check/DellEMC_SC_iSCSI/e3b5d2b/DellEMC_SC_iSCSI/189 : SUCCESS in 1h 30m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"b82336e30c14fdf221b22fadf96e58aa4c1bd798","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-07 16:33:10.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b58fbb84809509f2a1bc4d764c5cef0d762ccc8c","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-07 16:34:11.000000000","message":"Patch Set 5:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":5},{"id":"7c9896d5d9fa0a16ea1c3c0ca9ed5078c37e7be8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-07 17:12:46.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/767290/5/check/DellEMC_VNX_iSCSI/ff279e0/DellEMC_VNX_iSCSI/245 : SUCCESS in 2h 59m 30s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/767290/5/check/DellEMC_VNX_FC/a3d8b45/DellEMC_VNX_FC/246 : SUCCESS in 1h 11m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"ebdd11dffe838286fc36ff6f1316725100bb3ce9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-07 21:24:19.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"b4125862769c18df4d9beaee724596dfd7188ec0","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-08 01:03:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/5/20210108075033/NEC-ISCSI : SUCCESS in 2h 11m 40s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/5/20210108075033/NEC-FC : SUCCESS in 2h 13m 04s","accounts_in_message":[],"_revision_number":5},{"id":"4615000fdac1997b05b7c15124ff0543aa3c6efc","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-08 04:23:40.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/5/check/dsvm-tempest-TOYOU-ACS5000-driver/aa628e4 : ABORTED in 36m 31s","accounts_in_message":[],"_revision_number":5},{"id":"4bf47da003a7fd8852ecbd193484c5830952865d","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-08 04:25:03.000000000","message":"Patch Set 5:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":5},{"id":"0a27732efc5c46ac53aa1a73a3482d83c93d98f1","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-08 04:29:59.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/5/check/dsvm-tempest-TOYOU-ACS5000-driver/0bf9884 : FAILURE in 3m 29s","accounts_in_message":[],"_revision_number":5},{"id":"c9493850c9cb4ad2031df5900d6b4998505f4ad2","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-08 04:31:03.000000000","message":"Patch Set 5:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":5},{"id":"d4c78ec918e3157e015a3aa06fe4a0068028db23","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-08 04:49:22.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/5/check/dsvm-tempest-TOYOU-ACS5000-driver/caf9187 : FAILURE in 17m 00s","accounts_in_message":[],"_revision_number":5},{"id":"46d592b3201dd7a32b3f4aa72dad544c5319989b","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-08 04:53:23.000000000","message":"Patch Set 5:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":5},{"id":"d626979917d78d888f06f206f756d48bd22ff616","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-08 05:27:23.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"23def3b64a25fdc06e88e00d76b063089a2f88e9","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-08 06:54:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/5/check/dsvm-tempest-TOYOU-ACS5000-driver/143c74f : SUCCESS in 1h 59m 01s","accounts_in_message":[],"_revision_number":5},{"id":"26d71bfb85b261e1bfc717baa98ba77e2e59bf73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-08 08:11:10.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"f73538880bc3896888a504e64365b7e1f710dd3d","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-10 17:23:41.000000000","message":"Patch Set 5:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":5},{"id":"efed10afc89163e587e6814be9ec75bccac58da4","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-10 20:21:44.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/5/check/dsvm-tempest-TOYOU-ACS5000-driver/68df5a9 : FAILURE in 1h 31m 36s","accounts_in_message":[],"_revision_number":5},{"id":"6b93ddc5173c0725f6eb2d58d3182afa9df38593","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-10 21:43:52.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"055c4fc060e10258131766f6db2b7b31319aa18f","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-11 01:00:58.000000000","message":"Patch Set 5:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":5},{"id":"cefc75fee7f54e730ebdc8793d8ddb2f7e7b9963","author":{"_account_id":32925,"name":"lisa","email":"lisa.qiao@dell.com"},"date":"2021-01-11 02:30:36.000000000","message":"Patch Set 5:\n\nrun-DellEMC XtremIO CI","accounts_in_message":[],"_revision_number":5},{"id":"154da1093fbad22444a795b0ba459ec3158c64ce","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-11 03:57:16.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/5/check/dsvm-tempest-TOYOU-ACS5000-driver/b73e809 : FAILURE in 23s","accounts_in_message":[],"_revision_number":5},{"id":"f41036f9b0bb03af6a8b43293fe72bb8c7ae2c0a","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-11 04:00:01.000000000","message":"Patch Set 5:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":5},{"id":"530e9e35e658c986cdf41ad70cee1fe0117e96da","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-11 04:37:03.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/5/check/DellEMC_XtremIO_FC/057ec6f/DellEMC_XtremIO_FC/476 : SUCCESS in 2h 06m 03s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/5/check/DellEMC_XtremIO_iSCSI/9d11e4a/DellEMC_XtremIO_iSCSI/463 : SUCCESS in 1h 31m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"5eaed3feddae7e996b6be2888bb26040d1c33ead","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-11 05:14:42.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"17d225e4f2b79c41b03aec870fa9848bec107dc5","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-11 07:11:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/5/check/dsvm-tempest-TOYOU-ACS5000-driver/f3148db : SUCCESS in 1h 14m 38s","accounts_in_message":[],"_revision_number":5},{"id":"7372be1360a9988e773b0f5d4ca82173b4ff1a0f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-11 08:29:55.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e062865ca41f08d975508fe9b485f4afbdf27e56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-11 09:52:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1d0b30672d3f6670956081f31766421cfbaa0c11","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-11 09:52:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6c7272e40f11dfa6bd4a00a403b89693bb7a1e2a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-11 10:04:18.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera finger://openstackcifc-virtual-machine/c57fd46cdeb9423b88797120917a6371 : RETRY_LIMIT in 5s","accounts_in_message":[],"_revision_number":6},{"id":"8968e7e8e3d3cc1175f93f275490127edf7a449e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-11 10:25:27.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/6/check/dsvm-tempest-TOYOU-ACS5000-driver/75a34b5 : FAILURE in 31m 58s","accounts_in_message":[],"_revision_number":6},{"id":"29494cab5867e5ee0f790b13bfe8d9d3487af2d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-11 10:26:44.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"b59d8f484f955e725af05e5e60163bd34575394a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-11 10:46:04.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-6/ns5_nfs : FAILURE in 53m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-6/ns5_iscsi : FAILURE in 27m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"065357b941e71440d6f58136d4ccacfa721fbd48","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-11 10:49:44.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/6/check-not-vote/ext-cinder-zuul/6295f82 : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":6},{"id":"b4ff336348ed5d62dcba19f56373a968d75ea188","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-11 10:49:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/3d11659/3d11659.html : FAILURE in 55m 31s","accounts_in_message":[],"_revision_number":6},{"id":"365995172b0fdca035c93c4605538d0c2469e2e5","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-11 10:54:57.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/6/iscsi/ : FAILURE in 2m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/6/smb/ : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":6},{"id":"d532dbc56a94ec2d0869da01b5730ed87f81a1fd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-11 11:12:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/6/check/dsvm-tempest-cisco-zonemanager-job/2947ae7/ : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":6},{"id":"2e21f286f74f64ace0d02e58133482d110d704eb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-11 11:38:22.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"02a7ce7b10020fd660f2544385f12bcba97c2801","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-11 11:56:33.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/6/check/DellEMC_XtremIO_FC/b28a1ed/DellEMC_XtremIO_FC/477 : SUCCESS in 2h 04m 00s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/6/check/DellEMC_XtremIO_iSCSI/e71cbd6/DellEMC_XtremIO_iSCSI/464 : SUCCESS in 1h 31m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"8b379b834122bec0a5f0a789231ebbaa90ed47ee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-11 12:14:56.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/6/check/tempest-dsvm-full-focal-py3/7f6d028/ : SUCCESS in 1h 38m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/6/check/devstack-plugin-ceph-dsvm-focal-py3/86ac98b/ : SUCCESS in 2h 12m 13s","accounts_in_message":[],"_revision_number":6},{"id":"2bd62842ee76b9500ade7d54623a8f941c1e80e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:42:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a09af3609cc64960a6d93955ae6b7631 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d0eea29f4174bfeb01c714b55beddae : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/845cabcc5a8b4fb89aac73324ff16381 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b01277c2abfe40029c9d59782aee01e7 : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97aa90993a344b45a36356c681ae10d7 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f24a253dac945c387cbf2090a4f96b6 : SUCCESS in 15m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/8f467ea3aad047c7ad5e3144c33438f3 : SUCCESS in 1h 02m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ca97d70946c4a0e94f21d801f55bd5f : SUCCESS in 1h 19m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0870244fcf7545759a6b4fc89e726c9b : SUCCESS in 11m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2edcb3ebc0a24e6ab64b2ce1c4acf1b5 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/668cd1f571214ca4ba59a553e4ccd414 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e17fc1ea02ac428f936fc06319068021 : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9722f8367b3744ab8c9cd6df111a76d1 : SUCCESS in 1h 07m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d1e1b91934cf43a2b9593caa9414d62b : FAILURE in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e062ba292df4d319e8f07733994212a : SUCCESS in 1h 04m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/72632a9ff4d1481499850f721a066f44 : FAILURE in 1h 15m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8eace405fb284cf786b3095ec5c1e43a : FAILURE in 1h 24m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b24583562b3e4c05a2ff35def043faf3 : SUCCESS in 1h 09m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/66f1982b72c845fe9f10d4c9d9494592 : SUCCESS in 1h 21m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3bee9808d19a484faeff3d282b545027 : SUCCESS in 1h 28m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a88dde3008b148c0b5d94d25824c2c12 : SUCCESS in 2h 15m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b95aa49b31d645948fd5da4847059715 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":6},{"id":"9c440d1c6c4582ae8a9fa2d5c06122b7abee0ade","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-11 12:44:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/6/20210111185204/NEC-ISCSI : SUCCESS in 2h 28m 52s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/6/20210111185204/NEC-FC : SUCCESS in 2h 52m 32s","accounts_in_message":[],"_revision_number":6},{"id":"bdfa41e8115bb5478eb684e3f72460d4871752cb","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-11 13:22:28.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3b88a7213aa59736655d39db287bae75d8e6f9d7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-11 13:22:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"33697c705193f85fb8a620c87e810b0bef243a45","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-11 13:23:02.000000000","message":"Patch Set 6:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":6},{"id":"ea68e19607fa46d3a87ee6bf017d426d0449e7fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-11 14:14:23.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"465e716cba2c8a1a5e6917bd20d18cd8ebb1514d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-11 14:20:39.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/6/iscsi/ : FAILURE in 41s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/6/smb/ : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":6},{"id":"f23fc45be1d930a6b2afeafca42eb582cfe5f810","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-11 14:45:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/6/check/dsvm-tempest-cisco-zonemanager-job/fd0a07e/ : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":6},{"id":"d779c2429605faf787c5bbb32944c9114390608b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-11 14:46:00.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/767290/6/check/DellEMC_PowerMAX_FC/9503cb7/DellEMC_PowerMAX_FC/371 : SUCCESS in 4h 53m 27s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/767290/6/check/DellEMC_PowerMAX_iSCSI/fea7796/DellEMC_PowerMAX_iSCSI/341 : SUCCESS in 4h 32m 44s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"1be757d20b0463daa3ed96512ebf01455db0a88e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-11 15:27:32.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/6/check-not-vote/ext-cinder-zuul/0e8cd7e : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":6},{"id":"e2904025776c1729ac15f524947b775b53a60483","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-11 15:39:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/6/check/dsvm-tempest-TOYOU-ACS5000-driver/ffd238b : SUCCESS in 1h 32m 24s","accounts_in_message":[],"_revision_number":6},{"id":"24ef2bce187ceb03517ee2806224bcebfb88311c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-11 16:17:24.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/6/check/tempest-dsvm-full-focal-py3/bbaf249/ : SUCCESS in 1h 56m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/6/check/devstack-plugin-ceph-dsvm-focal-py3/defaf09/ : SUCCESS in 2h 17m 40s","accounts_in_message":[],"_revision_number":6},{"id":"2885563da61b8be6eb98eff73c21314fed9477d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-11 16:51:08.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"2c9fea70b709ada00b80cb3567488f6c1af074cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 17:41:15.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b40153b37ed74fafa565568e57c97e9c : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b87fbbd34b460b9f60594c4d7bed3b : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfc40485698341e2a04ec3a56f32b2e4 : SUCCESS in 12m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/589eedc4654040e58a696f8af8a83c0b : SUCCESS in 10m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a1d1e019150410eb40a4156657e908e : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c607dc3d71c427aaf360ca345e22b41 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/9e67b80e1e574f87abadf0ab634b8357 : SUCCESS in 1h 08m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b0785d830534d27a67308da75a2dfcd : SUCCESS in 1h 22m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7849e40a2f074ae69aadaa283d1b8483 : SUCCESS in 13m 02s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1d401b1b20c49bb8fb99cd483b1a69e : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5654e53df6d482dbcac23193fdb63a6 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b9c5ab0200094828953ec6d9a158879a : SUCCESS in 8m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8cca4101fae54006ae24400d89a047de : SUCCESS in 45m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49f4681af2ed4ee29bc0b3b76a140c60 : FAILURE in 9m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/65692562632a402ea933c9b9c539ce06 : SUCCESS in 1h 04m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f9f3cbc1370b4c099161f9d851d8a005 : FAILURE in 1h 14m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/51256a48a4a64a28bb83c101e180f3a3 : FAILURE in 2h 05m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79edabaa8f7c44c69238d450201e576e : SUCCESS in 1h 06m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7ebd53a5d65f4581914ea7e601d48e57 : SUCCESS in 1h 20m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/23af5f0aa1c4471f9ef5c70ec5218f20 : SUCCESS in 1h 20m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d7b9079af514fd4840cca7f4b8b59c2 : SUCCESS in 2h 11m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/403710548925465c9bccc86eb6849848 : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":6},{"id":"fb56612b686f740861ee0024839c2d74b5d3b1eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-12 03:04:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"17cf4cce900db2aedb9ce3af84654f2e6c5d7127","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-12 03:05:06.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"08ae9849c05d690db290d7b957c8bef7256179ec","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-12 03:37:00.000000000","message":"Patch Set 7:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":7},{"id":"385f56768b5a21e47470b460883278af4aa740e1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-12 04:01:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/7/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/dc8cfaf/dc8cfaf.html : FAILURE in 55m 22s","accounts_in_message":[],"_revision_number":7},{"id":"bfb8c94d0968ef1b7edb9d1ab5d9a941a963517d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-12 04:06:03.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-6/ns5_nfs : FAILURE in 6h 18m 12s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-6/ns5_iscsi : SUCCESS in 4h 33m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"a321f80f89e23912a7be8259885c5a75cfa85e74","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-12 04:09:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/7/iscsi/ : SUCCESS in 1h 03m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/7/smb/ : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":7},{"id":"2738c85c2a90df892db0d14fa99e662381cf2939","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-12 04:17:53.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/7/check/dsvm-tempest-TOYOU-ACS5000-driver/72b1c9f : FAILURE in 35m 33s","accounts_in_message":[],"_revision_number":7},{"id":"cf74bdc7f47063cdf111039b216102c13b1c164b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-12 04:26:00.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/7/check-not-vote/ext-cinder-zuul/9fc8c02 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":7},{"id":"8735386e1fbd84505694c5d497d4616b04790d04","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-12 04:27:42.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/7/check/dsvm-tempest-cisco-zonemanager-job/4089b5f/ : SUCCESS in 1h 18m 09s","accounts_in_message":[],"_revision_number":7},{"id":"be2ef69ab9aa26e200f44d9accc06367dde6187b","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-12 04:45:23.000000000","message":"Patch Set 7:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":7},{"id":"7393ffea44ffdec75449fd1fcad394c165a877c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 05:06:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76e6583237fb427ab22936c379e2916d : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288a1926b0d74238a1ccaac967399b07 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/259f6c3b39e0492297b72b66a689339b : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e41475b50d564104a7869d18877df2cb : SUCCESS in 8m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c3b23bb94774fde874e5f5ea7ae6815 : FAILURE in 12m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f9f9e5b3f947f88f24e649430305fe : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f85238f450164cb6b427f8ae48632eb1 : SUCCESS in 51m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70d97a22e7e3486ab4cf6ac0d1a94d2d : SUCCESS in 1h 17m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b86b568cf0ef4de2b9a94817b893c58b : SUCCESS in 7m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/71a61ec5e2db4c7d8aa90fd384433b34 : SUCCESS in 15m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/12d5a22c1c864152a6902cd0f2798e9a : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c07b12567204328bfdf6c96956fce3b : SUCCESS in 6m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fb3ca4c30314c51b63eb7217a24c8f1 : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/36fc293ac1674b0783843f5917319e0e : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a7b48ac6837c4ba7ac17bf2c361fe4a0 : FAILURE in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d3155abbbca64a7d8d6cdd5058bd7fa4 : SUCCESS in 1h 04m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4a4149158bdd4be7b709fa44941c2c03 : FAILURE in 1h 26m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/021915edb1a5408d95fc23f4bd144356 : SUCCESS in 1h 33m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc3f31cb115c4ebb870ce304c935fc4d : SUCCESS in 1h 08m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1c48fc28947045cd89a26f84324a94a5 : SUCCESS in 1h 33m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2d0ee35658b94c62ab0b89b3b26a5d55 : SUCCESS in 1h 32m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a390304f3c2547efae53bfa9cf4c8e15 : SUCCESS in 2h 00m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac817beaa3fa430a8df139df3f2f9f97 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":7},{"id":"ad8f458826cbc88a2577dee4843e7001d320c596","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-12 05:10:03.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/7/check/tempest-dsvm-full-focal-py3/e964d05/ : SUCCESS in 1h 51m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/7/check/devstack-plugin-ceph-dsvm-focal-py3/59f1ea1/ : SUCCESS in 1h 40m 19s","accounts_in_message":[],"_revision_number":7},{"id":"c9a60b5a8b4b8569cc44da8b3c4a2ed039206ff7","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-12 05:44:40.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/7/check/dsvm-tempest-TOYOU-ACS5000-driver/cad5ec3 : ABORTED in 57m 59s","accounts_in_message":[],"_revision_number":7},{"id":"7cec800fbeb43b50fa0cd07dfbf7097f07cbbcc3","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-12 05:45:59.000000000","message":"Patch Set 7:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":7},{"id":"5875c856d4226fd024882476d1285b348c2982cc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-12 06:03:52.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/7/check/DellEMC_PowerStore_iSCSI/79fac91/DellEMC_PowerStore_iSCSI/253 : SUCCESS in 56m 55s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/7/check/DellEMC_PowerStore_FC/a15e455/DellEMC_PowerStore_FC/247 : SUCCESS in 50m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"bf52072a550ed967f3762473a769b7b40bc2cbd2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-12 06:11:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/7/upstream-check/cinder-cDOT-iSCSI/661b9b5/ : SUCCESS in 37m 17s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/7/upstream-check/cinder-cDOT-iSCSI-IPv6/7ed8db6/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/7/upstream-check/cinder-cDOT-NFS/b31484c/ : SUCCESS in 38m 13s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/7/upstream-check/cinder-cDOT-NFS-IPv6/8d04791/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/7/upstream-check/cinder-cDOT-FCP/10b3485/ : SUCCESS in 1h 03m 55s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"6c08f746a92bd5d9a90bbbdad2292efb5f457eee","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-12 07:15:03.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-7/ns5_nfs : FAILURE in 55m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-7/ns5_iscsi : SUCCESS in 3h 48m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"3ac5e129e6fb8f4c3d6734d539f45cb3bf50a374","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-12 08:01:37.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/767290/7/upstream-solidfire-check/cinder-solidfire-iSCSI/f4e3aef/ : SUCCESS in 53m 28s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"6054f17a1072b173481ec30fed56e1faff990db3","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-12 08:16:28.000000000","message":"Patch Set 7:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/7/check/dsvm-tempest-TOYOU-ACS5000-driver/999b42f : SUCCESS in 1h 21m 03s","accounts_in_message":[],"_revision_number":7},{"id":"45f1932f58087087cd71f090042a7b6b32cdedec","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-12 08:20:53.000000000","message":"Patch Set 7:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-7/Seagate-iSCSI : SUCCESS in 45:31.42\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-7/Seagate-FC : SUCCESS in 40:56.60","accounts_in_message":[],"_revision_number":7},{"id":"bd020a77b4f8256d6c34d72879892733c04872e6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-12 08:48:48.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/767290/7/check/DellEMC_SC_FC/31dc4c6/DellEMC_SC_FC/198 : SUCCESS in 2h 07m 07s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/767290/7/check/DellEMC_SC_iSCSI/73a362c/DellEMC_SC_iSCSI/198 : FAILURE in 1h 50m 19s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"ef4efbee0012ba6705071c41fedc59f938538bf9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-12 09:12:58.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/7/check/DellEMC_PowerFlex/5afed65/DellEMC_PowerFlex/210 : SUCCESS in 2h 16m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"25c93403dc167b7b6e953821b0b7c4b4c1e95dc8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-12 11:06:36.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/7/check/DellEMC_XtremIO_FC/7d42ed8/DellEMC_XtremIO_FC/488 : FAILURE in 2h 04m 10s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/7/check/DellEMC_XtremIO_iSCSI/62c7549/DellEMC_XtremIO_iSCSI/475 : SUCCESS in 1h 33m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"8b609bb170e148ef75c64020af825344d15f8619","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-12 12:17:45.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/7/20210112185216/NEC-ISCSI : SUCCESS in 2h 12m 03s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/7/20210112185216/NEC-FC : SUCCESS in 2h 25m 26s","accounts_in_message":[],"_revision_number":7},{"id":"aa8226ec96ca0b3ba9cfca7daf08a78e08766dd6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-12 13:39:03.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"44f235de97dd1650c15ec51d46bda5d98bf7acbd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-12 16:03:18.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-7/ns5_nfs : SUCCESS in 3h 20m 24s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-7/ns5_iscsi : SUCCESS in 3h 18m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"1b52b2dfdba463efc470761c71b87d4111efbb73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-12 16:58:27.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/90/767290/7/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"a99fd01c6cf51ef08bc226f5b6a556ebce6ffbe4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-12 18:44:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/b5cb1ee90de540a0a213eb69e5aa5d8d : FAILURE in 14m 43s","accounts_in_message":[],"_revision_number":7},{"id":"d6ef1b45306357379cd2da97e6523d41db267567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-14 01:58:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b226dfd7220646afa01029c7b33615da544758f7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-14 01:59:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"718bdca3d2778115de5dea823d2a076aa715e9fe","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-14 02:02:32.000000000","message":"Patch Set 8:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":8},{"id":"29fa58cc3998edd6076e8d93e227d03dd3d2faa6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-14 02:18:11.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/287a7bb004fb4cbd9a494451ad2304f8 : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":8},{"id":"2acd8b520bef567ece0822f9dddaa271efc0d8b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-14 02:32:41.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5c9a16612f6eada25b70c06ccca583c4a12f01e5","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-14 02:33:08.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3b9260441059a8b1b0cef0664b337ba09dedd4aa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-14 02:44:51.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/9/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c5c6d06/c5c6d06.html : FAILURE in 6m 51s","accounts_in_message":[],"_revision_number":9},{"id":"3142cdb8968e94c80b752e4ffc0852bc33d45b75","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-14 02:52:10.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/0ba39cf33c794d3aaea29cd8550fff95 : FAILURE in 14m 51s","accounts_in_message":[],"_revision_number":9},{"id":"ad5eb814587b3c95a08e5de111a640c450a28d59","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-14 03:16:35.000000000","message":"Patch Set 9:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":9},{"id":"129c93b3831d675280e4f70bbae122823aba9feb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-14 03:37:05.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/9/iscsi/ : SUCCESS in 1h 03m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/9/smb/ : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":9},{"id":"938de5ec90032025ae6eb24b9cbbf99f4858f9c1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-14 03:54:55.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/9/check/dsvm-tempest-cisco-zonemanager-job/2bf4d1f/ : SUCCESS in 1h 17m 15s","accounts_in_message":[],"_revision_number":9},{"id":"db5169a6c4f0516a70b7b09ae13e3bdd5fe065fe","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-14 04:18:28.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/9/check/dsvm-tempest-TOYOU-ACS5000-driver/4ab899b : FAILURE in 34m 13s","accounts_in_message":[],"_revision_number":9},{"id":"cbad766ef4922a06024736dbd463abcd71e0f10a","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-14 04:25:24.000000000","message":"Patch Set 9:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":9},{"id":"0705e26a839a7c077993b4ff81409c887ce65641","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-14 04:32:54.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/9/check/tempest-dsvm-full-focal-py3/01eb0b3/ : FAILURE in 1h 45m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/9/check/devstack-plugin-ceph-dsvm-focal-py3/d27703c/ : FAILURE in 1h 46m 05s","accounts_in_message":[],"_revision_number":9},{"id":"c7d096c8952c91af310d09a7db959b8e337d3a79","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-14 05:10:33.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-8/ns5_nfs : SUCCESS in 3h 11m 10s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-8/ns5_iscsi : SUCCESS in 3h 9m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"c90bdcc64ef3e374369169841186d542962dfca6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-14 05:59:06.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-9/ns5_nfs : SUCCESS in 3h 25m 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-9/ns5_iscsi : SUCCESS in 3h 6m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"005a440dab7fab6be7b88d1c2913741591df8650","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-14 06:59:41.000000000","message":"Patch Set 9:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/9/check/dsvm-tempest-TOYOU-ACS5000-driver/3c13f61 : SUCCESS in 1h 17m 39s","accounts_in_message":[],"_revision_number":9},{"id":"248a4491625f995fa9dcf1e37cc070b2c613b6f6","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-14 07:22:46.000000000","message":"Patch Set 9:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":9},{"id":"ae2bb4297445641c66b66cdc097b1c50b5a8c812","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-14 07:46:08.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/9/check/dsvm-tempest-TOYOU-ACS5000-driver/99974fc : FAILURE in 21m 52s","accounts_in_message":[],"_revision_number":9},{"id":"2922c535fac11d32233f97980d0350aa38fb4fbe","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-14 07:47:09.000000000","message":"Patch Set 9:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":9},{"id":"e1ea2d4278ffdc3a01f81604cf65e62f23df001a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-14 09:03:12.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/9/check/DellEMC_XtremIO_FC/80dfb66/DellEMC_XtremIO_FC/507 : SUCCESS in 2h 02m 42s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/9/check/DellEMC_XtremIO_iSCSI/2ad44b0/DellEMC_XtremIO_iSCSI/497 : SUCCESS in 1h 32m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":9},{"id":"24fbfafaf0ea72775d86acea9d8564d269c65255","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-14 09:13:27.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/9/20210114155211/NEC-FC : SUCCESS in 2h 13m 59s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/9/20210114155211/NEC-ISCSI : SUCCESS in 2h 21m 14s","accounts_in_message":[],"_revision_number":9},{"id":"6b8497fd249dca4ef16374913691448feddd3be9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-14 10:18:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1685f8d0d12c1c1c9627ffe3d6262da5c91dbb3e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-14 10:18:36.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cd57b024f859c074199322f2b8a815cf2c69dab9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-14 10:37:45.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci https://ci.open-e.com/zuul/t/local/build/9412dca01acb43c8ac3633554562ec5b : FAILURE in 14m 49s","accounts_in_message":[],"_revision_number":10},{"id":"d23caf19c1648cfa305d78fb44d8917f54922893","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-14 11:32:53.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a50f141/a50f141.html : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":10},{"id":"322d0d79c95b0d0323b633255bf80e714bdb4096","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-14 11:33:12.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/10/iscsi/ : SUCCESS in 1h 09m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/10/smb/ : FAILURE in 1h 14m 15s","accounts_in_message":[],"_revision_number":10},{"id":"23f6b6a5499fc54ded696e0249c0a57b72a1462b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-14 11:47:39.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/10/check/dsvm-tempest-cisco-zonemanager-job/2e63a0b/ : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":10},{"id":"9b620db44aadc797a1823870bbc6346ba8f2c645","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-14 12:55:36.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/10/check/tempest-dsvm-full-focal-py3/bf7301b/ : FAILURE in 1h 51m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/10/check/devstack-plugin-ceph-dsvm-focal-py3/3f8d7a7/ : FAILURE in 2h 09m 46s","accounts_in_message":[],"_revision_number":10},{"id":"4640cd7feea3c2762af60756636154a1f0dd87c3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-14 13:26:56.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-10/ns5_nfs : SUCCESS in 3h 8m 11s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-10/ns5_iscsi : SUCCESS in 2h 58m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"ae43dda0d7113042f5930f6ee5117d207cd3a854","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-14 13:56:16.000000000","message":"Patch Set 10:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":10},{"id":"0b5e4c1db13e8c899543cb2b1b6b33c4e42b8d44","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-14 14:00:32.000000000","message":"Patch Set 10:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":10},{"id":"d225b6c98fef88dba6cb45d341d811474501c254","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-14 15:39:35.000000000","message":"Patch Set 10:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/10/check/dsvm-tempest-TOYOU-ACS5000-driver/1d838ce : SUCCESS in 1h 22m 38s","accounts_in_message":[],"_revision_number":10},{"id":"1623b4b02b76e589760483b827ea77d4bc893894","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-14 20:26:58.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/10/check/DellEMC_XtremIO_FC/5799802/DellEMC_XtremIO_FC/512 : SUCCESS in 2h 04m 29s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/10/check/DellEMC_XtremIO_iSCSI/7a6454a/DellEMC_XtremIO_iSCSI/502 : SUCCESS in 1h 32m 41s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"e22be92fc891248d10c822d1a24f6da093c4aced","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-14 23:12:57.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/10/20210115033950/NEC-ISCSI : SUCCESS in 4h 18m 48s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/10/20210115033950/NEC-FC : SUCCESS in 4h 33m 04s","accounts_in_message":[],"_revision_number":10},{"id":"15b47f84221bf3c0ec59a6d7169eb726e747c8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 05:03:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10ba45b1fe544908b81dc9f770ad26d8 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaec7a3741c0457a9830f5b3b3a38c1d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9601ff8700143718115b080aa3f01c2 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf974b5f992412c93f8c356c988e8e1 : SUCCESS in 9m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3ef8920e55e450f820119e27ff27937 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86e159fb141455d8688a6dfb0d0145a : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8e0e829b1ea149e4bce6a117fa7c6ea9 : SUCCESS in 56m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75a52964f24a4ce9b56ddf87718e2efa : SUCCESS in 1h 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38c857b7ee7e4875bb567709d47581be : SUCCESS in 5m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fb1b73e4d8cb4b92a9ab6e82e270fe06 : SUCCESS in 12m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a83deb502ee241fbbc899f1004f9f80d : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8597042845a046b7b9fb721497a09f5f : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97ea56dc1c374e96bf4092c265c0a34f : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d67dd6d0b9f4198ae5c18c38338dd9f : SUCCESS in 35m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c465b123a2f249c1a733c234dc41b880 : FAILURE in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/06763a712a7444cb85d915a7067c6f09 : SUCCESS in 1h 17m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f8bc28b6deeb4261b0ed2b9113ca42ac : FAILURE in 1h 16m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/716d069f13cd4f44a0013928bcc56477 : FAILURE in 1h 57m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae4bde64637c4ec2812a61fb1ef5f91a : SUCCESS in 1h 16m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e742669c1cc54a4eabab268d01f972b6 : SUCCESS in 1h 31m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/773f9b38235d428e89a8bcfe34361ded : SUCCESS in 1h 20m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b24433a19284a82b640eb597f3bfdc3 : SUCCESS in 2h 03m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afc7bc204e5040afb89161ddc7b75ae6 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":10},{"id":"e86309982cee34e58cbf11211be055815af7888b","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-15 05:40:56.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"643999ea779ca80d1f23eb2a9d0a6e2243e28d02","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-15 05:41:21.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ad5c478287e7216f2e67732e20236d07ee7a4be4","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-15 06:42:17.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/10/iscsi/ : SUCCESS in 1h 00m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/10/smb/ : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":10},{"id":"b7efcbdc6321e338af4049b58eff3426a28f110e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-15 07:03:31.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/10/check/dsvm-tempest-cisco-zonemanager-job/43c3d4f/ : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":10},{"id":"14ce2b0a790e9a7ee5c80dc0b43e136ae8fe3ba1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-15 07:25:32.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ae4e54a79fb8442aacaff748922805e4 : RETRY_LIMIT in 1h 22m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/10/check/devstack-plugin-ceph-dsvm-focal-py3/0352a2d/ : SUCCESS in 1h 30m 21s","accounts_in_message":[],"_revision_number":10},{"id":"9eadfb1894175aae77f8796f446b83ae0311157b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-15 13:28:53.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/10/check-not-vote/ext-cinder-zuul/49e6c4a : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":10},{"id":"88d59fbde7adc67bae44622e76d9991f519a3aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:39:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c05ce2d8a51641fd9ad043a955cb7c3d : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b38cf28d63d549c388a4d0c4ddc9f142 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bef8d3a53095482fa7a9e56df4bac03d : SUCCESS in 11m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f21eaeeeee714789a59f4e79efdbe9cd : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bce65a68a253499899923327a995d442 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74eae1beae1c48e290e7ffc5c43dca76 : SUCCESS in 15m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/b169c98573b346029d9ba5da235a062d : SUCCESS in 1h 14m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f70da65a3814304914156c96f4df9f0 : SUCCESS in 1h 17m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5da6ee138474710a3e2647430cd019c : SUCCESS in 5m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0a68de80335d45b3bf58d8fb95dc47a8 : SUCCESS in 16m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2d0e98a1d8748d0bdafb7daa2739578 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/762293538863478ebc88202b77e9d842 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3f9fdee03ae54e30b23d2ceb317e92f6 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/948c52153d1247c7b597f5450dd5ac41 : SUCCESS in 49m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98e8cb0db1054161bb23da833ff7c509 : FAILURE in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/299cff04f668487f86f15bfe2ed45e3c : SUCCESS in 1h 09m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/49123ad58d1749c68d68d8b3f2eefb3d : FAILURE in 1h 12m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fcd534045fe84700a45d0e9473198ec0 : SUCCESS in 1h 33m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc95d71bf75a4bed868311749f57a279 : SUCCESS in 1h 18m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4b3200bbbd5d4694a3a1a9d5399e0837 : SUCCESS in 1h 32m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6c7dae76ef714ddc9df29394419b055a : SUCCESS in 1h 33m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c664a51db7f40bebd0abf48c7f9df73 : SUCCESS in 2h 35m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53ed385882e64e21bf68a1bfb1713db4 : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":10},{"id":"a35d7b7a6eda20413bd39d526c56048e51844145","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-01-15 20:41:56.000000000","message":"Patch Set 10: Code-Review+1\n\n(10 comments)\n\nThere are things I would like to see fixed.  Not giving it a -1 for what is here.","accounts_in_message":[],"_revision_number":10},{"id":"11a44f7272423fa1cf568d9a4a372cf58ebf34cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-15 21:06:49.000000000","message":"Patch Set 10: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":10},{"id":"21be83520fe79ca2adca8799510170069abfd483","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-15 22:01:39.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/10/upstream-check/cinder-cDOT-iSCSI/18c3da5/ : SUCCESS in 39m 19s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/10/upstream-check/cinder-cDOT-iSCSI-IPv6/e8b98d0/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/10/upstream-check/cinder-cDOT-NFS/c3701d0/ : SUCCESS in 39m 44s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/10/upstream-check/cinder-cDOT-NFS-IPv6/290392c/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/10/upstream-check/cinder-cDOT-FCP/177805d/ : SUCCESS in 1h 20m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"4995dd791fbfdac18337bcf163cbebeeadb3dc12","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-15 22:25:17.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/10/check/DellEMC_PowerStore_iSCSI/2b40e9a/DellEMC_PowerStore_iSCSI/266 : SUCCESS in 57m 11s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/10/check/DellEMC_PowerStore_FC/ab56f3b/DellEMC_PowerStore_FC/260 : SUCCESS in 50m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"1c3da07d2407364b32bd39ac9db5e6c6e5d3bbf8","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-15 22:26:49.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/767290/10/upstream-solidfire-check/cinder-solidfire-iSCSI/4082fa4/ : SUCCESS in 1h 45m 28s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":10},{"id":"189d0370a474c948e6f5ccfaa2b01a0c7d2b3472","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-16 02:01:19.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/90/767290/10/check/DellEMC_PowerMAX_FC/9bcbad9/DellEMC_PowerMAX_FC/389 : SUCCESS in 6h 04m 17s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/90/767290/10/check/DellEMC_PowerMAX_iSCSI/8ba14e6/DellEMC_PowerMAX_iSCSI/354 : SUCCESS in 6h 16m 20s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":10},{"id":"86cfbb589eedfad733330fafc9557da4ce838a37","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-16 03:03:19.000000000","message":"Patch Set 10:\n\n\u003e Patch Set 10: Code-Review+1\n\u003e \n\u003e (10 comments)\n\u003e \n\u003e There are things I would like to see fixed.  Not giving it a -1 for what is here.\n\nThanks for your review. I\u0027m fixing these unproper things...","accounts_in_message":[],"_revision_number":10},{"id":"df00b438051ed7a55e0b59231ae4924c7269d1f6","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-16 03:16:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a31f0fb8b81807643189d77c0987f31a8c2976a6","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-01-16 03:31:22.000000000","message":"Patch Set 10:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/767290-10/PVME-iSCSI : SUCCESS in 52:48.57\n* PVME-FC http://os-logs.tristero.net/dell-ci/767290-10/PVME-FC : SUCCESS in 49:31.91","accounts_in_message":[],"_revision_number":10},{"id":"11054cda61cc3c01c7c044558f0be4493be5b94d","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-16 05:09:27.000000000","message":"Patch Set 10:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-10/Seagate-iSCSI : SUCCESS in 49:07.76\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-10/Seagate-FC : SUCCESS in 45:24.54","accounts_in_message":[],"_revision_number":10},{"id":"2b5e8e5a19dd19cf460cdd433c23ca4a42b3be6b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-16 05:34:57.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/10/check/DellEMC_Unity_FC/1c4bbb7/DellEMC_Unity_FC/446 : SUCCESS in 1h 25m 24s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/10/check/DellEMC_Unity_iSCSI/04a49ed/DellEMC_Unity_iSCSI/411 : SUCCESS in 1h 46m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"792539b95482a38b53d81ee60f9ccda5fedbcd1a","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-01-16 05:38:23.000000000","message":"Patch Set 10:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/767290-10/Lenovo-iSCSI : SUCCESS in 48:26.18\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/767290-10/Lenovo-FC : SUCCESS in 43:45.12","accounts_in_message":[],"_revision_number":10},{"id":"6a5363fc2c82a4cdb496087a2cd44dc7bdc5e549","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-16 05:57:14.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/767290/10/check/DellEMC_SC_FC/35706ce/DellEMC_SC_FC/211 : FAILURE in 2h 21m 32s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/767290/10/check/DellEMC_SC_iSCSI/5bba4eb/DellEMC_SC_iSCSI/211 : FAILURE in 1h 22m 47s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"371fb3764731a8fd24c8188169e2889cc5459a1e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-16 06:07:38.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/10/check/DellEMC_PowerFlex/b943ae0/DellEMC_PowerFlex/223 : SUCCESS in 2h 16m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"2de7240be1b0f92abde6044f70216bb380a52cca","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-01-16 07:44:41.000000000","message":"Patch Set 10:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/767290-10/HPMSA-iSCSI : FAILURE in 1:07:54\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/767290-10/HPMSA-FC : SUCCESS in 40:50.74","accounts_in_message":[],"_revision_number":10},{"id":"0d548f98196202b13a090332fd5bcb8394c5d0a7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-16 08:13:16.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/767290/10/check/DellEMC_VNX_iSCSI/ec91697/DellEMC_VNX_iSCSI/262 : SUCCESS in 2h 37m 34s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/767290/10/check/DellEMC_VNX_FC/027d1fe/DellEMC_VNX_FC/261 : SUCCESS in 1h 24m 18s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"8ac6dc4d2d8b5ff017a66b435872ef121a4bf14e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-16 22:43:22.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/10/check/DellEMC_XtremIO_FC/e895977/DellEMC_XtremIO_FC/536 : SUCCESS in 2h 05m 55s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/10/check/DellEMC_XtremIO_iSCSI/eab82aa/DellEMC_XtremIO_iSCSI/527 : SUCCESS in 1h 35m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"a9783e634e1b74ab4c611649acdf4e958c2936b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-18 18:00:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"16874ca09cf349b2665d39a55f841c645b5ebee7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-18 18:00:51.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3a8416ad6ed716529a2c75d8b25827f0a5107651","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-18 18:12:32.000000000","message":"Patch Set 11:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":11},{"id":"32bec1f60f2dc853301d88bcc052087a96652888","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-18 18:15:10.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/11/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/af4474d/af4474d.html : FAILURE in 12m 30s","accounts_in_message":[],"_revision_number":11},{"id":"8cbd0b02eaafb16c3f051e53eba7539797f888da","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-18 19:19:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/90/767290/11/check/iscsi-driver-primera/2154db2/2154db2.html : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":11},{"id":"686e999fed2fa2cc6944aa181b46b723efc1bcc3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-18 19:21:41.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/11/check/dsvm-tempest-cisco-zonemanager-job/ac65d1d/ : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":11},{"id":"21e922f1e525ad0eb6317dc184b567e69ce2297e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-18 20:10:53.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/11/iscsi/ : FAILURE in 1h 24m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/11/smb/ : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":11},{"id":"5352e171b1c85a371663600240095b6d4d3ca69e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-18 20:15:58.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/11/check-not-vote/ext-cinder-zuul/80a97fb : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":11},{"id":"276f6520b524142158abd1dd1315e65d60c3694d","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-18 20:19:55.000000000","message":"Patch Set 11:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/11/check/dsvm-tempest-TOYOU-ACS5000-driver/6a553ee : SUCCESS in 1h 15m 02s","accounts_in_message":[],"_revision_number":11},{"id":"8b25903ec87ae3afe21935610e44b71fd53066c4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-18 20:44:08.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-11/ns5_nfs : SUCCESS in 2h 38m 0s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-11/ns5_iscsi : SUCCESS in 2h 43m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"74c4af37aab0559e6e07f34426c5dfe12dae2109","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-18 22:24:41.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/11/check/DellEMC_Unity_FC/44811eb/DellEMC_Unity_FC/470 : SUCCESS in 1h 26m 49s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/11/check/DellEMC_Unity_iSCSI/d35b6f5/DellEMC_Unity_iSCSI/435 : FAILURE in 1h 47m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"cb335e9dd53145c42ba4b018bf947b5b584a4415","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-18 22:52:09.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/11/check/tempest-dsvm-full-focal-py3/05f6839/ : SUCCESS in 1h 39m 31s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/11/check/devstack-plugin-ceph-dsvm-focal-py3/edc2ab6/ : SUCCESS in 1h 47m 36s","accounts_in_message":[],"_revision_number":11},{"id":"7edec38ab8fce0f093fc1882551c5a6c150c62f3","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-18 23:00:13.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/11/20210119053631/NEC-FC : SUCCESS in 2h 04m 54s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/11/20210119053631/NEC-ISCSI : SUCCESS in 2h 23m 38s","accounts_in_message":[],"_revision_number":11},{"id":"e6684df554cf66c985c295bd12260f3c15502e28","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-18 23:31:06.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/11/check/DellEMC_XtremIO_FC/9873aa6/DellEMC_XtremIO_FC/552 : SUCCESS in 2h 04m 40s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/11/check/DellEMC_XtremIO_iSCSI/3567169/DellEMC_XtremIO_iSCSI/543 : SUCCESS in 1h 35m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":11},{"id":"b0963145d1e831841cfb32452b516492c550ed01","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-18 23:55:35.000000000","message":"Patch Set 11: Code-Review+1\n\n(3 comments)\n\nA few nits noted inline.\n\nJust want to note that the CI has passed on this patch (though I don\u0027t think it\u0027s running the cinder-tempest-plugin).  It also has run on at least one unrelated patch https://review.opendev.org/c/openstack/cinder/+/562053 as is required, though there are also some patches that it hasn\u0027t responded on.  As far as CI goes, I think we\u0027re OK for the purposes of merging this patch and CI responsiveness and including cinder-tempest-plugin tests can be addressed before Milestone 3.\n\nI\u0027ll let Sean and Jay re-review the code since they\u0027ve looked carefully already.","accounts_in_message":[],"_revision_number":11},{"id":"8dc2148de45b7a544aaf82f2acf9b39f5646fabd","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-19 02:16:59.000000000","message":"Patch Set 11:\n\n\u003e Patch Set 11: Code-Review+1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e A few nits noted inline.\n\u003e \n\u003e Just want to note that the CI has passed on this patch (though I don\u0027t think it\u0027s running the cinder-tempest-plugin).  It also has run on at least one unrelated patch https://review.opendev.org/c/openstack/cinder/+/562053 as is required, though there are also some patches that it hasn\u0027t responded on.  As far as CI goes, I think we\u0027re OK for the purposes of merging this patch and CI responsiveness and including cinder-tempest-plugin tests can be addressed before Milestone 3.\n\u003e \n\u003e I\u0027ll let Sean and Jay re-review the code since they\u0027ve looked carefully already.\n\nBrian Rosmaita: Thanks for your review. \n1.About \"cinder-tempest-plugin\": \nI have done this: \"export \u0027PROJECTS\u003dopenstack/cinder-tempest-plugin \u0027\" for every CI process.\nBut I have to find the reason why it doesn\u0027t work between \"Milestone 2\" and \"Milestone 3\".\n\n2. About CI responsiveness:\nDue to my driver code is not merged yet. If I set CI response for event \"patchset-created\". \nzuul merger can\u0027t create a \"ZUUL_REF\" which contained my driver code. \nSo I set CI only runs on \"third party CI recheck trigger\". I don\u0027t know this set is ok or not.\n\n3.About \"host assisted\"\nI read \"https://docs.openstack.org/cinder/latest/reference/support-matrix.html#required-driver-functions\" for reference.\nWhen running cmd \"cinder help migrate\", it appears a message: \"cinder migrate \u003cvolume\u003e \u003chost\u003e\". \n\"host\" means \"Destination host\". Takes the form: \"host@backend-name#pool\".\nThe \"host\" parameter is required for migration. But if it is not proper here, I can delete it.","accounts_in_message":[],"_revision_number":11},{"id":"9303acd5d8ed23467adb04d0e444307f4e04669f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 02:17:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d88c4cac4ff747fe841b1c1597bb8b44 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac050ede83ea4e13b9185808ffc48afd : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2150a06266e45f5839066f00f25ca46 : SUCCESS in 14m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5721a965b5641b981cb5fb286527ae3 : SUCCESS in 15m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80909f3d7ddc475a89da66f85bcdaee7 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89df56f7259a43e09b694ae94d58a303 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/07b1f60b3ae74a11b3a330e01b01b512 : SUCCESS in 1h 02m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b93ae4c84e544d6bde61959322c86ca : SUCCESS in 1h 21m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/597e8cabb25d4753a5f8fd42c6fe0f8f : SUCCESS in 6m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/22693665de21487e91341875ac5f9055 : SUCCESS in 14m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d44ca357eaa45c4a6143ce8955e2a67 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/987670e134aa4985ab89b10b0c3f3d5e : SUCCESS in 11m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/119978c7c95f44f79857352a1b666245 : SUCCESS in 11m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/feeb24b7531046e89133b6e2919336e5 : SUCCESS in 50m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/641367e3ca9e443db6243ea29acdb56c : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/36c8632410324827a8f2d1a739806a51 : SUCCESS in 1h 10m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d34f73ef1624407d8c83bfe30168eacd : FAILURE in 1h 20m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1f773cf522a427d846d5805c8b8ab4c : SUCCESS in 1h 33m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7fe174ade1aa46e7bd1d68c391339d94 : SUCCESS in 1h 06m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dd40fec2421e472eb58fb4d6f6c0a1d1 : SUCCESS in 1h 39m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0627b0ae131348a4a748683eea1deac0 : SUCCESS in 1h 26m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2479279a23ac4a27be6e3fb8c9d28156 : SUCCESS in 2h 11m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80f43a5012b84316941f2063e14fae5d : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":11},{"id":"84319e55c3f288327562bd877997f839f21d423c","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-19 02:51:41.000000000","message":"Patch Set 11:\n\n(2 comments)\n\nThanks,I will solve it in the next patch.","accounts_in_message":[],"_revision_number":11},{"id":"3c199db5521165d01c040a5a260751c2f423ee64","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-19 03:14:04.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/11/check/DellEMC_PowerStore_iSCSI/dc3ff98/DellEMC_PowerStore_iSCSI/286 : SUCCESS in 51m 33s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/11/check/DellEMC_PowerStore_FC/14a5fa7/DellEMC_PowerStore_FC/280 : SUCCESS in 53m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"970ff50632d4ceb8fe41b3e6415c7271ecfc40d1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-19 03:35:46.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/11/upstream-check/cinder-cDOT-iSCSI/c8344eb/ : SUCCESS in 36m 38s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/11/upstream-check/cinder-cDOT-iSCSI-IPv6/47dbad6/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/11/upstream-check/cinder-cDOT-NFS/3a23877/ : SUCCESS in 36m 43s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/11/upstream-check/cinder-cDOT-NFS-IPv6/d813173/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/11/upstream-check/cinder-cDOT-FCP/b6d95ab/ : SUCCESS in 1h 16m 55s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"eb88d878d20d1fe283b4fad13457541fe21d5c40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-19 03:49:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ebc4bfce308f26ef0520ec2c8e1818479cbd4977","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-19 03:50:24.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"32222d5c38807052888d2d158cb8be8fe07a9b87","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-19 04:35:14.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/12/check-not-vote/ext-cinder-zuul/4b1f983 : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":12},{"id":"7b0eff64f568a5a66bc9ac90ff1c6bcaf2d26f7e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-19 04:51:12.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/12/iscsi/ : FAILURE in 47m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/12/smb/ : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":12},{"id":"fe77dfd33818c32bd19682b608f8013839d47826","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 05:01:01.000000000","message":"Patch Set 12:\n\nrun-HPE Storage CI","accounts_in_message":[],"_revision_number":12},{"id":"77855f0fb7ad93189add220a6e61e7c10c21e689","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 05:01:20.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c5f63a9/c5f63a9.html : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":12},{"id":"695cd2bce6e5ae42b82e9352ecc512da5aa768d9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 05:08:19.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/90/767290/12/check/iscsi-driver-primera/7fd76b0/7fd76b0.html : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":12},{"id":"bed9fbe1cbc15b47a65a0b5d2721525c7dde0e6b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-19 05:13:08.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/12/check/dsvm-tempest-cisco-zonemanager-job/52a7203/ : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":12},{"id":"7edf3c011db3c4aa0b2d816f0728372864212a2a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 05:32:18.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a494597/a494597.html : FAILURE in 29m 45s","accounts_in_message":[],"_revision_number":12},{"id":"f39e124e84e59bb3a98ba7b966fea625bec0b989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 05:32:48.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e07c645df994ab49dcb0795e4fc7579 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0e0a48532648fe9691e13f90674396 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c98d44e96a62415a9266b856b8a9411b : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d02b5d20c7b4664ab60a00f2f8c5c74 : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc2b878c91bb4a4387c7d43454fd00aa : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f2ab9f93a6454a96b35487d8bb85f7 : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/cfacfcb65c234e3f82af81c5830f319f : SUCCESS in 54m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/137a24bdc79348918be11cb97cded702 : SUCCESS in 1h 02m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6aa527089d044f4b886e38899d20edf : SUCCESS in 5m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ef212c0356154ba9a896635d8857587e : SUCCESS in 12m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/395b5052f15c4e438280810985ec0939 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f162b9fb82e542558b57903df116f070 : SUCCESS in 5m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a1f49c7503e4867bed905e0cbb4c409 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c89537009bac49d58a68026d2838b321 : SUCCESS in 35m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/982b56e03b3f4f76b0df352c12091606 : SUCCESS in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/239d1c8d6c70484e95b169b0acaa3f7d : SUCCESS in 57m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b9a96c5ccf5841aa820f8ed5c81e34a8 : FAILURE in 1h 15m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf0a7db6a91b4a3688df952352f26a50 : SUCCESS in 1h 28m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/545e6fc21401441d9436047ae209bc75 : SUCCESS in 59m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cc91703aa2db4ea2804c20c3e559043b : SUCCESS in 1h 13m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eb5ac9560f6e4e98a929aa9b4b4a2b84 : SUCCESS in 1h 25m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4b7b26e256b7427484d400deb5c2b1f6 : SUCCESS in 1h 39m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/337085145c314e409281cc838810bc1c : SUCCESS in 51m 08s","accounts_in_message":[],"_revision_number":12},{"id":"2f8f680b8454cf8f12461d7d785f3a64a59a4248","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-19 05:41:39.000000000","message":"Patch Set 12:\n\n(28 comments)\n\nI have replied to the review suggestions of patch10 and patch11","accounts_in_message":[],"_revision_number":12},{"id":"9246c49719d28e176ae4ae7e672550d07e55dd63","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-19 05:49:39.000000000","message":"Patch Set 11:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-11/Seagate-iSCSI : SUCCESS in 51:25.53\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-11/Seagate-FC : SUCCESS in 48:19.00","accounts_in_message":[],"_revision_number":11},{"id":"85f7e88ef2ae5aaf026288290dbbae49fd6195d9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-19 05:53:32.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/12/check/tempest-dsvm-full-focal-py3/97c57c0/ : SUCCESS in 1h 43m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/12/check/devstack-plugin-ceph-dsvm-focal-py3/ca88f43/ : SUCCESS in 1h 55m 27s","accounts_in_message":[],"_revision_number":12},{"id":"0d33994e06036b2dfc74fb35e9cd3fa105509a97","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-01-19 05:55:42.000000000","message":"Patch Set 11:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/767290-11/Lenovo-iSCSI : SUCCESS in 49:48.61\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/767290-11/Lenovo-FC : SUCCESS in 49:39.10","accounts_in_message":[],"_revision_number":11},{"id":"fbdbe8860b8c033273e793e3f266332bb46070a6","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-19 06:21:58.000000000","message":"Patch Set 12:\n\n(4 comments)\n\nreply to the review suggestions of patch4","accounts_in_message":[],"_revision_number":12},{"id":"0315212eacdfb3e4a2942c892e24083b5a27d815","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-19 06:29:12.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/12/check/DellEMC_PowerStore_iSCSI/57019fd/DellEMC_PowerStore_iSCSI/288 : SUCCESS in 55m 32s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/12/check/DellEMC_PowerStore_FC/57ed2be/DellEMC_PowerStore_FC/282 : SUCCESS in 51m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"dc0446d010f73a143167985862ac504415327c8d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-19 06:33:32.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-12/ns5_nfs : SUCCESS in 2h 38m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-12/ns5_iscsi : SUCCESS in 2h 43m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"30715e6e93c1aaeb637fc431766ebb4db7bd600f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 07:16:25.000000000","message":"Patch Set 12:\n\nrun-HPE Storage CI","accounts_in_message":[],"_revision_number":12},{"id":"470d51047a4be8b745485bc0132f2cfc0fa2349b","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-19 07:28:59.000000000","message":"Patch Set 12:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":12},{"id":"be40b1cf47e141f2f30806a71bcdccec4f03b439","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 07:33:34.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e1b7443/e1b7443.html : FAILURE in 15m 08s","accounts_in_message":[],"_revision_number":12},{"id":"9e8dd34c4ce82de24defca99e932b9a95bc5373a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-19 08:23:26.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/767290/12/check/DellEMC_SC_FC/3ee2899/DellEMC_SC_FC/233 : SUCCESS in 2h 13m 59s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/767290/12/check/DellEMC_SC_iSCSI/b5a6458/DellEMC_SC_iSCSI/233 : FAILURE in 1h 35m 39s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":12},{"id":"295d8ca625ecabad3a570cf16e40b74adaa53222","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-19 08:24:12.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/12/upstream-check/cinder-cDOT-iSCSI/0d6fc02/ : SUCCESS in 35m 26s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/12/upstream-check/cinder-cDOT-iSCSI-IPv6/13d1225/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/12/upstream-check/cinder-cDOT-NFS/5ca2e7f/ : SUCCESS in 38m 56s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/12/upstream-check/cinder-cDOT-NFS-IPv6/f42ecb5/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/12/upstream-check/cinder-cDOT-FCP/7f0341b/ : SUCCESS in 1h 08m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"c485e7b8f09c2c3497da817f147a5e2e06b3bc39","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 08:26:34.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/12/20210119132154/NEC-FC : SUCCESS in 2h 18m 26s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/12/20210119132154/NEC-ISCSI : SUCCESS in 4h 04m 37s","accounts_in_message":[],"_revision_number":12},{"id":"14d0c45bc7aeb587366b8a9019cb9899a985fb82","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-19 08:40:48.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/767290/12/upstream-solidfire-check/cinder-solidfire-iSCSI/116ba0f/ : SUCCESS in 54m 50s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"e1c44fef364e80a952fe9492e706343db242fcf3","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-19 08:45:53.000000000","message":"Patch Set 12:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/12/check/dsvm-tempest-TOYOU-ACS5000-driver/fe62252 : SUCCESS in 1h 14m 44s","accounts_in_message":[],"_revision_number":12},{"id":"be97d4e3231ba14bb035bdc0e57d718944755f5a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-19 09:00:52.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/12/check/DellEMC_Unity_FC/7a90ff9/DellEMC_Unity_FC/476 : SUCCESS in 1h 28m 32s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/12/check/DellEMC_Unity_iSCSI/372f7ff/DellEMC_Unity_iSCSI/441 : SUCCESS in 1h 38m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"02399b15dde564fb6255421abd0dcf7396541a63","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 09:33:47.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/12/check/DellEMC_PowerFlex/ea5a241/DellEMC_PowerFlex/243 : ABORTED in 4h 00m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"1d73bdef9955a7da90585fa949c3a04e4c22ae71","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 09:53:43.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/767290/12/check/DellEMC_VNX_iSCSI/f4e7381/DellEMC_VNX_iSCSI/283 : SUCCESS in 2h 42m 52s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/767290/12/check/DellEMC_VNX_FC/ebbe3be/DellEMC_VNX_FC/283 : SUCCESS in 1h 13m 16s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":12},{"id":"6f4160775821d228778ef59e47cdc05034c4d2c6","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-19 10:19:20.000000000","message":"Patch Set 12:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-12/Seagate-iSCSI : SUCCESS in 49:40.63\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-12/Seagate-FC : SUCCESS in 45:06.10","accounts_in_message":[],"_revision_number":12},{"id":"3d4b3d4ce246b4b92ec18e38888ae110108d5ff5","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-01-19 10:38:11.000000000","message":"Patch Set 12:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/767290-12/HPMSA-iSCSI : FAILURE in 1:08:54\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/767290-12/HPMSA-FC : SUCCESS in 35:25.12","accounts_in_message":[],"_revision_number":12},{"id":"4d883e966bbe754c6fc4b0a4b72d61550ca786ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-19 10:51:32.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"248db44cff1b368c9f02cafb07f967b9d33a909f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-19 10:52:12.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"91acfe1b9a67efe083f9148593bd812f80ff1928","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 11:20:49.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a3a08da/a3a08da.html : FAILURE in 1h 31m 16s","accounts_in_message":[],"_revision_number":12},{"id":"bd36c7461b2a7f8c6b97572fe1a55935f30be534","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 11:28:10.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/902720e/902720e.html : FAILURE in 16m 24s","accounts_in_message":[],"_revision_number":13},{"id":"9d1e0c899cb230ca13b34bfa56c9211bee62586e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-19 11:32:31.000000000","message":"Patch Set 13:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/13/check-not-vote/ext-cinder-zuul/53b10fd : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":13},{"id":"a824e5ef739b3224b5c7c2439308dd2b399b6d46","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 11:34:06.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0df87cc/0df87cc.html : FAILURE in 2m 27s","accounts_in_message":[],"_revision_number":12},{"id":"c6948fffc5e864bd367ac3f208af4cca1c0d14c0","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-19 11:38:54.000000000","message":"Patch Set 13:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":13},{"id":"3a835ac60ad83d0c9a7c19f904811fa86d534380","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 11:55:38.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/eb30024/eb30024.html : FAILURE in 2m 19s","accounts_in_message":[],"_revision_number":12},{"id":"97b1d7e62256e6054ddfeff571c5e73232084287","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 11:55:53.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/524d7ea/524d7ea.html : FAILURE in 2m 19s","accounts_in_message":[],"_revision_number":13},{"id":"f6166891a5108abd3f965ce4d331060fefdb0351","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 12:12:00.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/90/767290/13/check/iscsi-driver-primera/6ef2d8f/6ef2d8f.html : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":13},{"id":"3dd39e1a29a7dbcc7b836c0dbc2250e9576b78c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 12:27:08.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/7d9a1cf/7d9a1cf.html : FAILURE in 2m 35s","accounts_in_message":[],"_revision_number":13},{"id":"938937b2f91b053220020884e37fd892de808cc1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-19 12:39:58.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/13/check/DellEMC_Unity_FC/6c0a365/DellEMC_Unity_FC/477 : FAILURE in 13m 35s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/13/check/DellEMC_Unity_iSCSI/c19c282/DellEMC_Unity_iSCSI/442 : SUCCESS in 1h 47m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"cbd04179d4e28888f60531633b17d96471ed6d43","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-19 12:41:21.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/13/iscsi/ : SUCCESS in 1h 15m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/13/smb/ : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":13},{"id":"7b39e70a995c61cd55503cb033fa504573b6e31a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 12:58:07.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/13/check/DellEMC_XtremIO_FC/5498085/DellEMC_XtremIO_FC/559 : FAILURE in 2h 05m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/13/check/DellEMC_XtremIO_iSCSI/4357cb6/DellEMC_XtremIO_iSCSI/550 : SUCCESS in 1h 35m 57s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":13},{"id":"dbc97cb28e303e004b8df60c99949c83db390c13","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-19 13:25:26.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/13/check/dsvm-tempest-cisco-zonemanager-job/22854c6/ : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":13},{"id":"ccef081edd9e918c54de060925a61a5b300a885e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-19 13:42:02.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-13/ns5_nfs : SUCCESS in 2h 47m 55s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-13/ns5_iscsi : SUCCESS in 2h 49m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"179751cbac3b6d28aece1ea48494d09d7f161295","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 13:42:19.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/18a2a29/18a2a29.html : FAILURE in 1h 11m 42s","accounts_in_message":[],"_revision_number":13},{"id":"78a6f31839c3127946ed7baf209313030fdc6a70","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-19 14:00:28.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/12/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1482154/1482154.html : FAILURE in 1h 20m 34s","accounts_in_message":[],"_revision_number":12},{"id":"67e42e51914c3229cb25b93e36da30dd1da0226d","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-19 15:22:23.000000000","message":"Patch Set 13:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/13/check/dsvm-tempest-TOYOU-ACS5000-driver/91f7e55 : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":13},{"id":"714b96f0a62b914a35ba42bf8d2a503c0ea8d8e7","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 15:29:05.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/13/20210119215306/NEC-FC : SUCCESS in 2h 25m 27s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/13/20210119215306/NEC-ISCSI : SUCCESS in 2h 35m 56s","accounts_in_message":[],"_revision_number":13},{"id":"7ba6b8628e3095e02e34872af8ca974e07518844","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-19 15:35:56.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 11: Code-Review+1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e A few nits noted inline.\n\u003e \n\u003e Just want to note that the CI has passed on this patch (though I don\u0027t think it\u0027s running the cinder-tempest-plugin).  It also has run on at least one unrelated patch https://review.opendev.org/c/openstack/cinder/+/562053 as is required, though there are also some patches that it hasn\u0027t responded on.  As far as CI goes, I think we\u0027re OK for the purposes of merging this patch and CI responsiveness and including cinder-tempest-plugin tests can be addressed before Milestone 3.\n\u003e \n\u003e I\u0027ll let Sean and Jay re-review the code since they\u0027ve looked carefully already.\n\nHi, Brian Rosmaita. I have added cinder_tempest_plugin test into my CI process. And the succeed log url has been uploaded.","accounts_in_message":[],"_revision_number":13},{"id":"72494e50eb1fd02ef7121223d84feab03e3bbff6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-19 17:21:56.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/13/check/tempest-dsvm-full-focal-py3/fa2c1fd/ : SUCCESS in 2h 00m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/13/check/devstack-plugin-ceph-dsvm-focal-py3/dfff585/ : TIMED_OUT in 2h 42m 35s","accounts_in_message":[],"_revision_number":13},{"id":"0c5dcc5bb7c7021a50879e3a9ef7bcf2215cd862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 19:51:37.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dba74c03051548ae99df62234315f417 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07bb42d75a61492eb9fae910a1ba0aeb : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66914ac3dcf247e3b6a7f2d86abd7cdf : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba5ecc26cb0e4f85ac21c299a2038b7d : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f2f994fb7d14b85a30c914801ea919b : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa3b9ecfb06437f96860c19232079d4 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/d1933b3dba464eefb30bd9e81de962cd : SUCCESS in 1h 11m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60042f1fb6cf4b279e11005c6eec5ecd : SUCCESS in 1h 08m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a9253e794d94d5dac853c668958dc99 : SUCCESS in 7m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b3d07bc4474b4ec4bc99cbb8f9d0391d : SUCCESS in 12m 42s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef7ee6c9fb8f4a3992aeacaa6c8dc5d9 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f47539282a504105bf94a93cd3b6e6bf : SUCCESS in 6m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/653c074286d54778bd5c3cfd9ad902f2 : SUCCESS in 9m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9781ca04f62149bbadf5dd9c3f1cdd3b : SUCCESS in 39m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42c2ae8953614f30baa2894f5f567d9c : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d22c8097b89c4a7b9628c373d9380a12 : SUCCESS in 1h 13m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/41c3a467b6fb4384863a199e3f385f51 : FAILURE in 1h 14m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6be3f599ceb84d18823e9a9b53b7489e : SUCCESS in 2h 11m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c0e7a08271304178b057e80f9da1dc15 : SUCCESS in 1h 31m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3d7750ec55a5471ba6e84d347a5c1fd1 : SUCCESS in 1h 37m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/da272c66caf2412e81aafa543b49aac0 : SUCCESS in 1h 14m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b52ab9f4eb53412aa77ab41b024bc590 : SUCCESS in 2h 13m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04aea2fdc28c4dad8addad8a9098579a : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":13},{"id":"c9301249145527342e92b722f193728ad8cfcaa1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-01-19 19:58:47.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)\n\nThanks for addressing my other comments.  We need to make sure to remove the use of six.text_type before this can merge.","accounts_in_message":[],"_revision_number":13},{"id":"a6b57d0fde12287220fd120548df597ce2beed83","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-19 20:46:56.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/13/check/DellEMC_PowerStore_iSCSI/d268478/DellEMC_PowerStore_iSCSI/289 : SUCCESS in 54m 37s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/13/check/DellEMC_PowerStore_FC/ff08941/DellEMC_PowerStore_FC/283 : SUCCESS in 49m 08s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"82b1f2820d9c20eda8038dc3163a3c79bad54bcf","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-19 22:04:26.000000000","message":"Patch Set 13:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-13/Seagate-iSCSI : SUCCESS in 51:18.40\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-13/Seagate-FC : SUCCESS in 43:20.72","accounts_in_message":[],"_revision_number":13},{"id":"a1998cd27aafdc4155d8d6b0ea2d42fc06b836c7","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-19 22:05:10.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/767290/13/check/DellEMC_SC_FC/6881460/DellEMC_SC_FC/234 : SUCCESS in 2h 12m 57s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/767290/13/check/DellEMC_SC_iSCSI/12fc113/DellEMC_SC_iSCSI/234 : FAILURE in 1h 35m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":13},{"id":"5771f3a3fab2bc88477a2c40cb7f8effbed340a0","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-01-19 22:11:58.000000000","message":"Patch Set 13:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/767290-13/HPMSA-iSCSI : FAILURE in 1:03:59\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/767290-13/HPMSA-FC : SUCCESS in 38:03.30","accounts_in_message":[],"_revision_number":13},{"id":"6792575f3d3de31c0c54f050b07b6ddb5d60cf89","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-01-19 22:16:21.000000000","message":"Patch Set 13:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/767290-13/Lenovo-iSCSI : SUCCESS in 53:28.76\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/767290-13/Lenovo-FC : SUCCESS in 52:00.18","accounts_in_message":[],"_revision_number":13},{"id":"21191edcf4416e21e4e9af13bb506724282ac26c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 22:31:45.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/767290/13/check/DellEMC_VNX_iSCSI/816fdcb/DellEMC_VNX_iSCSI/284 : SUCCESS in 2h 39m 30s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/767290/13/check/DellEMC_VNX_FC/d90ebbf/DellEMC_VNX_FC/284 : SUCCESS in 1h 16m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":13},{"id":"e2669a226832bc6c6e3ad5f87ea20df5ae10030f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-19 23:52:26.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/13/check/DellEMC_PowerFlex/48d8605/DellEMC_PowerFlex/245 : ABORTED in 4h 00m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"f0e6f81a6e0992f13d3175d5ab83fa502fbbeac3","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-20 01:27:24.000000000","message":"Patch Set 13:\n\n(2 comments)\n\n\u003e Patch Set 13: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Thanks for addressing my other comments.  We need to make sure to remove the use of six.text_type before this can merge.\n\nThank you for reminding. We will make changes on patch14","accounts_in_message":[],"_revision_number":13},{"id":"569b61edb425d76cdac245ebeadec5ebe04f6a1b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-20 06:17:00.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/13/check/DellEMC_Unity_FC/3c5e42c/DellEMC_Unity_FC/486 : SUCCESS in 1h 31m 02s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/13/check/DellEMC_Unity_iSCSI/eca9d68/DellEMC_Unity_iSCSI/451 : SUCCESS in 1h 57m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"4ceac41588adf087ffcd6bbde9f3fa2eac1d1cf0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-20 08:35:45.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/13/check/DellEMC_XtremIO_FC/d00d4a1/DellEMC_XtremIO_FC/568 : FAILURE in 2h 05m 03s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/13/check/DellEMC_XtremIO_iSCSI/0d265ec/DellEMC_XtremIO_iSCSI/559 : SUCCESS in 1h 35m 09s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":13},{"id":"992a59cca0603f0dceaa2b83a43509fd04acb35e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-20 09:43:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"e3102261651b48e002e1c1993cfc6150e3272efd","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-20 09:43:57.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"90b4d3bd437ac99dffff280e59d3354bb676a6ce","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-20 09:47:23.000000000","message":"Patch Set 14:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":14},{"id":"fb0f26058c6f59dced37dcbc478d82d9651802ce","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-20 09:50:02.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/14/check/dsvm-tempest-TOYOU-ACS5000-driver/b55ac41 : ABORTED in 52s","accounts_in_message":[],"_revision_number":14},{"id":"93229641e05bf48f319723bbea0e86e0b08c62b5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 09:52:20.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/14/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/9af9ccb/9af9ccb.html : FAILURE in 7m 19s","accounts_in_message":[],"_revision_number":14},{"id":"ec3288340efdaf06a554a705977739ae5047fb77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-20 10:03:50.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3cc3b4051020d1e6fb4ded1b8d4be01ed313c796","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-20 10:04:27.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"64aef0d7f07ba3d921a413402930c2a5f6618dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-20 10:18:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"4aaf4dbf53de0cd1d75c983e6f7b04dcfc9e1df3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-20 10:18:55.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b9fc869c31faf49f98aab377c1e448c3365e31c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 10:32:15.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/90/767290/16/check/iscsi-driver-primera/b7ae05b/b7ae05b.html : FAILURE in 9m 04s","accounts_in_message":[],"_revision_number":16},{"id":"023fdb3d15e93aba4ba312d895306a934b84ef23","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-20 10:50:40.000000000","message":"Patch Set 16:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":16},{"id":"b7017ad52b4b434cf018bf75e6b922d7c2e2f5b0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-20 11:04:04.000000000","message":"Patch Set 16:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/16/check-not-vote/ext-cinder-zuul/313bc44 : SUCCESS in 44m 36s","accounts_in_message":[],"_revision_number":16},{"id":"867b41ed731b9c5ce548ba8a1e69a729d33a3c3a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 11:06:21.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/16/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/12d8fcb/12d8fcb.html : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":16},{"id":"959d7caff625a3bf49b0f79217b9e02cfba44c92","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-20 11:41:29.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/16/check/dsvm-tempest-cisco-zonemanager-job/bf36a63/ : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":16},{"id":"97ab66df452f3015f074e78b95fa4178a747c29a","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-20 11:46:45.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/16/iscsi/ : FAILURE in 1h 27m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/16/smb/ : FAILURE in 1h 03m 29s","accounts_in_message":[],"_revision_number":16},{"id":"34fce954f1078bfc40f6763cfc03b8950d6a4196","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 11:57:08.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/16/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d9597f4/d9597f4.html : FAILURE in 25m 40s","accounts_in_message":[],"_revision_number":16},{"id":"fa03142148feacc68efaafebada0c58c2c0681c9","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-20 12:27:40.000000000","message":"Patch Set 16:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/16/check/dsvm-tempest-TOYOU-ACS5000-driver/6773827 : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":16},{"id":"797706a3f30a0017e9729c2d8ab38158b79c6844","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 12:35:17.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/90/767290/16/check/iscsi-driver-primera/985bc83/985bc83.html : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":16},{"id":"31b0b58912ff7c1fb9fb9e37599a70cbe6ae46fd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 12:38:19.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/14/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/5112ab5/5112ab5.html : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":14},{"id":"6657f2f40bc37daebe019358e774d5739123a78d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 12:55:14.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/16/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0f57d1e/0f57d1e.html : FAILURE in 7m 37s","accounts_in_message":[],"_revision_number":16},{"id":"64b2f58f00dba836777155dc84e82ee7333b6635","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-20 13:29:51.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-14/ns5_nfs : SUCCESS in 3h 40m 33s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-14/ns5_iscsi : SUCCESS in 3h 27m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"0e42e0457bdc5e7668e7f0cc4e0cdb78f8e17657","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-20 13:35:24.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-15/ns5_nfs : SUCCESS in 3h 30m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-15/ns5_iscsi : SUCCESS in 3h 27m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"523b4e74fce760589a27cc005b9cf2c76d7e6cb3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-20 13:47:18.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-16/ns5_nfs : SUCCESS in 3h 27m 50s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-16/ns5_iscsi : SUCCESS in 3h 25m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":16},{"id":"2658e0cbecd78ac6ca3b7f96cdab5fae66121dc8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-20 13:47:54.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/bcf99e1a8130472794041ea194c30043 : RETRY_LIMIT in 31m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/16/check/devstack-plugin-ceph-dsvm-focal-py3/a9b97e5/ : SUCCESS in 1h 48m 11s","accounts_in_message":[],"_revision_number":16},{"id":"35f1e922c8b63b252c69fb5f91d75a15da67b366","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 14:04:52.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/16/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/b3c190e/b3c190e.html : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":16},{"id":"95ce8e88d18b1279ac8fdd7444f74d0ae8cea8d3","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-01-20 15:09:58.000000000","message":"Patch Set 16: Code-Review+2\n\nMy comments have been addressed and it looks like the CI has passed.  I am ok with this.  Thank you for addressing the comments!","accounts_in_message":[],"_revision_number":16},{"id":"ad4ffe5a81d968faef8b8d3612524859ada0a725","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-20 15:29:27.000000000","message":"Patch Set 16:\n\n\u003e Patch Set 16: Code-Review+2\n\u003e \n\u003e My comments have been addressed and it looks like the CI has passed.  I am ok with this.  Thank you for addressing the comments!\n\nThanks a lot.","accounts_in_message":[],"_revision_number":16},{"id":"b6e8d0c406cdc0b40df3afb3264edfe1fac9cbe6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-20 15:48:27.000000000","message":"Patch Set 16: Code-Review-1\n\n(7 comments)\n\nBiggest issues are force delete and deleting a volume that no longer exists. Otherwise mostly minor things inline.","accounts_in_message":[],"_revision_number":16},{"id":"bee933f1c46e2b0e4908dac3fd5da48c65871f13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-20 18:29:46.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"0d503343733e1d38cbfd169dd2e321051a6338c5","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-20 18:30:30.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6a23a3bbb0eaeb55085d8b4281482908f1896cfd","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-20 18:33:08.000000000","message":"Patch Set 17:\n\n(7 comments)\n\nThe review suggestion has been corrected and the new patch has been uploaded.","accounts_in_message":[],"_revision_number":17},{"id":"f5ce4be706fec62e7ddf39edf5a371363fe8a389","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 18:39:47.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/17/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2235f5b/2235f5b.html : FAILURE in 8m 00s","accounts_in_message":[],"_revision_number":17},{"id":"8c28e22c6085ee51dfba495ee3cab9459848638c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-20 19:11:51.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/17/check-not-vote/ext-cinder-zuul/b6eae24 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":17},{"id":"8301e13322a9a2d12958ad6d9d11da258fbe033f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-20 19:46:22.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/90/767290/17/check/iscsi-driver-primera/42f2dff/42f2dff.html : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":17},{"id":"f3f822ad8030467429b8aab41c0f9862a010c8a9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-20 20:06:37.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/17/iscsi/ : FAILURE in 1h 35m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/17/smb/ : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":17},{"id":"24e87bb47baf298dca70959c0414473beb1cf169","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-20 20:21:19.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8b7c7c2ce0feda1034882907f0e0fdf25e816e2c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-20 20:43:00.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/17/check/tempest-dsvm-full-focal-py3/c05ee57/ : FAILURE in 1h 48m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/17/check/devstack-plugin-ceph-dsvm-focal-py3/b3d5c9b/ : FAILURE in 1h 17m 44s","accounts_in_message":[],"_revision_number":17},{"id":"0546a998efdc703bab77683aefdd65888522b5e4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-20 20:48:55.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/17/check/dsvm-tempest-cisco-zonemanager-job/5ce673f/ : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":17},{"id":"424d12b1b00e2d3b036be114fcbad9a9dda013e1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-20 21:26:22.000000000","message":"Patch Set 17:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-17/ns5_nfs : SUCCESS in 2h 55m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-90-767290-17/ns5_iscsi : SUCCESS in 2h 50m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":17},{"id":"06b2fa80b3a84527908aca1bac59817878996dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 00:51:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff5c275190dd49bdb577241e53657c36 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee220dfaee934d61998d0633b29d074e : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea0c7b201cac4f1586d97dc1af433fea : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7882b93a04d460caf55e246a0f35f46 : SUCCESS in 9m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5730bc4589048788b98565357f3fbbc : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa42abf5a2a4a54a5e310347ea53275 : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/d24ca2832c474c75a867989c6016bc68 : SUCCESS in 1h 02m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bdd4b64af2914dc9a92bfc8942e894c8 : SUCCESS in 1h 19m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9c13ea2762e4f7ea70142ce17a18f0b : SUCCESS in 6m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3da8f55ef41e4ab186b1fde0f065ad57 : SUCCESS in 12m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e4f3cfcaff341db94659f0469588ba4 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/557d991a72044f45a2cba469e3ed6d5a : SUCCESS in 6m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/217c1481ce73440fa80f488386b97cad : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0da864a9c5e04a5998e1f5e118518c94 : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b1012ea1a1740ca8e3869c63140d8e4 : FAILURE in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/184c15a0e95d41d88988d798446b793a : SUCCESS in 1h 09m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a357eea774d349baa7e9881500c6b1d4 : FAILURE in 1h 21m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d022cb9a2eb1467ebbd1a212a249dbea : SUCCESS in 1h 29m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/39c6a81661ad4df781217d18c25cee82 : SUCCESS in 1h 12m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b2fca4e98384d37bcb0a433b08efc82 : SUCCESS in 1h 14m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a12d96efb19946d0b4e9b21735aa4be5 : SUCCESS in 1h 42m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e5278eb93e443919f209e804bc0d3f0 : SUCCESS in 2h 14m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fbefc3af9b64ef2999c438d4d6740bd : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":17},{"id":"6f41cff4e165af622c46e359e95493e91c18c30d","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-21 01:06:42.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"2e6382db5c15c28f9b688dfa49c5a5e081e7c845","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-21 01:07:00.000000000","message":"Patch Set 17:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":17},{"id":"8004920129ce77a7f16dce59a2630526424b146b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-21 01:07:21.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"fd488062bcb8188f8dc97f774133235a888eea91","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-21 01:19:27.000000000","message":"Patch Set 17:\n\n(1 comment)\n\nReply to the target.","accounts_in_message":[],"_revision_number":17},{"id":"8cce8dea7a736076b3da7b9f897884d6d9b9d267","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-21 01:52:31.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/17/check-not-vote/ext-cinder-zuul/e77834f : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":17},{"id":"fb5dc4504b0eff73ac5ca5d968cc19c583e50383","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-21 02:01:56.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-iSCSI/420f019/ : SUCCESS in 37m 50s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-iSCSI-IPv6/e8c0173/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-NFS/86347be/ : SUCCESS in 37m 32s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-NFS-IPv6/4ff786c/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-FCP/1cc6a3b/ : SUCCESS in 1h 09m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":17},{"id":"f15c0d995ca53494ffd4bb0a609d91db997650f7","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-21 02:12:36.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/17/iscsi/ : SUCCESS in 1h 05m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/17/smb/ : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":17},{"id":"7f47adcfc8be598620572a85f2c743bf074394ce","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-21 02:19:04.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/767290/17/upstream-solidfire-check/cinder-solidfire-iSCSI/eef3a9f/ : FAILURE in 1h 25m 51s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":17},{"id":"9a2459c407ea2aa0b1e75e1a6cc01b3225c06848","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-21 02:19:14.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/17/check/dsvm-tempest-TOYOU-ACS5000-driver/d9601d4 : FAILURE in 1h 10m 17s","accounts_in_message":[],"_revision_number":17},{"id":"a6c27b848f37cf64320f7ae40066e035738b6df2","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-21 02:29:58.000000000","message":"Patch Set 17:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":17},{"id":"bbd46be881050e6571f2b4bcafe51f3a3d7cf8a8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-21 02:30:40.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/17/check/dsvm-tempest-cisco-zonemanager-job/e98bcca/ : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":17},{"id":"d93917738fa11eb4c5502ee9f599ca78f6452862","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-21 02:34:11.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/17/check/tempest-dsvm-full-focal-py3/57380d9/ : FAILURE in 1h 07m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/17/check/devstack-plugin-ceph-dsvm-focal-py3/f3531cb/ : FAILURE in 1h 20m 15s","accounts_in_message":[],"_revision_number":17},{"id":"c1120c7499a9d75a5d20fc129f09b5a99b993889","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-21 03:01:25.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/17/check/DellEMC_PowerStore_iSCSI/8e7e232/DellEMC_PowerStore_iSCSI/307 : FAILURE in 57m 10s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/17/check/DellEMC_PowerStore_FC/2fd7c6b/DellEMC_PowerStore_FC/301 : FAILURE in 50m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":17},{"id":"b321dbdc5dd64f17d4b54e52700ae2252ce4b2fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 03:15:06.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bdfcf9af97246b88d58793ac504edb0 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db931e184ed4253840b6220da8d2196 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68e602f243f146e1a9d729184362abef : SUCCESS in 12m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85131b2f9f1a4d009b1f4630dcaaf2a5 : FAILURE in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6584061b02a648478840fbc64ba91fec : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f118e3c5a3498aafc3095c60a54e73 : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/89911f5e3a884c32bae612300cf990ff : SUCCESS in 46m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35a529fbda714f7a86fcafa547bc0df7 : FAILURE in 1h 28m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cf4588866134d47b6ea07ef559c0086 : SUCCESS in 6m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/37b6755254ff4e91915e5a3d6aa54eb4 : SUCCESS in 11m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/753099201eba42fba4a02d7b43a85513 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/233ea8bb52374d92b2a214cf84382e8c : SUCCESS in 6m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d4dc306ee74c4e07ae5aa82fda8312e2 : SUCCESS in 9m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e63748884e8c4a9ca37f431edddc63ae : SUCCESS in 43m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/344ac97fea384f7d988a89344acde25e : FAILURE in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7450087921d0445584aea6487bc2630e : SUCCESS in 1h 07m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01d08e2321ce46c9bf5598689e91ff26 : FAILURE in 1h 16m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/24276f690c1045e6ac13e6222c0e099d : FAILURE in 2h 05m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1553364426ea485980e7902226775045 : SUCCESS in 1h 12m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fd4055143d4c489091ffc370d22ac0a8 : SUCCESS in 1h 11m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/42b44cf112b2460f9d2ee0ed46a42165 : SUCCESS in 1h 38m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8312007a9f5e487a92c9f1e0c81f3c6a : SUCCESS in 1h 57m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c9ecceeb1b44baa8294b75db5c44033 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":17},{"id":"b3395a594f2c2a50101bee3f51a2afd747441df3","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-21 03:45:41.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/17/check/dsvm-tempest-TOYOU-ACS5000-driver/d2bc353 : FAILURE in 1h 14m 04s","accounts_in_message":[],"_revision_number":17},{"id":"7950a08b12d23d01b4260b65baf2720c24956d4d","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-21 03:55:31.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"dfb6adc71e185e64f8fa2159dec09384c0028f8e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-21 03:56:04.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e3dc9dff3ededc828f32f81304cec700410fa568","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-21 04:36:44.000000000","message":"Patch Set 17:\n\nrun-TOYOU ACS5000 CI","accounts_in_message":[],"_revision_number":17},{"id":"a624bf59f7cad16a1b87a41649482a1a77b8ec7e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-21 04:40:02.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/17/check-not-vote/ext-cinder-zuul/7dd9699 : SUCCESS in 43m 35s","accounts_in_message":[],"_revision_number":17},{"id":"6c4e59f58bfe48a3362ebe50f2a25f7602512e43","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-21 05:14:45.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/17/check/tempest-dsvm-full-focal-py3/b59adf6/ : FAILURE in 1h 12m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/90/767290/17/check/devstack-plugin-ceph-dsvm-focal-py3/8cea059/ : FAILURE in 1h 08m 48s","accounts_in_message":[],"_revision_number":17},{"id":"b511771e025151b33c33091d1d363af7ec877042","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-21 05:15:24.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/17/iscsi/ : SUCCESS in 1h 03m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/17/smb/ : FAILURE in 1h 19m 08s","accounts_in_message":[],"_revision_number":17},{"id":"b0fe089bcd30700f23cece876aa07ac3292d92b8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-21 05:20:45.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/17/check/dsvm-tempest-cisco-zonemanager-job/0a4d182/ : SUCCESS in 1h 19m 51s","accounts_in_message":[],"_revision_number":17},{"id":"044da461f68270336fcad23a1ac18572e10cab1c","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-01-21 05:58:59.000000000","message":"Patch Set 17:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66/90/767290/17/check/dsvm-tempest-TOYOU-ACS5000-driver/bd86972 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":17},{"id":"775362d23a4d1fd5b7d4d32eb2ec705ab938c7c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 06:14:11.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62f42581d2734bc39b35eddccf7e937f : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00514447abff4e26adad3c5c6ce4487e : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f418af9a64ba484ab43fa943f244f3d5 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c6e92c1ebf0460fa31d75626d53511e : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60524293086f46eea8c76aeda78c1371 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de7da6b66c04336adbb2c6fb10a0678 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/00835a861c9e4bda934e4d5d6ed5b208 : SUCCESS in 55m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8524b0e048846988fa395fb7c34aa94 : FAILURE in 1h 02m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/556ac828801d4272a7f9f014a463293e : SUCCESS in 7m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ee64c599326d43ad9d7ed04696ee366b : SUCCESS in 10m 59s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31e5fa0523144f7c91d72c1c555e6173 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/769953dc2f904baf8c20bf98d586ea74 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63d641da65b0480eb3cb917763dbfea5 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eafa2d04e509428ea26451f938d47e89 : SUCCESS in 35m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/380bc4d6ebfb4f5885c783f4cd436b8d : FAILURE in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/644f89d6a03544b5984297101a9c0929 : SUCCESS in 1h 12m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/206498b4ecf8453285f11556835ba28e : FAILURE in 1h 16m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f4e78048a01a47e9be43bcd04f56a6be : FAILURE in 1h 16m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b394eaf13df346d692354f29a87a8dec : SUCCESS in 1h 03m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4f8c4fd88d3c43fe85eae000b43789ff : SUCCESS in 1h 07m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8a20896554634db9b15a66513703f13f : SUCCESS in 1h 15m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f2351b4317640cfa9b23b24b4f0e725 : SUCCESS in 2h 13m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/129d566322914828be728cdd667e7696 : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":17},{"id":"bb47f11add5701a9ccbff31fa8d1e722c5c302d0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-21 06:15:30.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/13/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/3ea1894/3ea1894.html : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":13},{"id":"4b08ca78fcb51c66fb42999892b06f90fca6e194","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-21 06:16:03.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/90/767290/17/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/f05b7d2/f05b7d2.html : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":17},{"id":"4a273a2019a8a46c12fdee5b4e261946152a59d5","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-21 06:29:21.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"412bd272d6f21dae5c458f849b4014b28eafb668","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-21 06:30:07.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f235cbb3ca548b071f56f43f6a2eb76e3f758dfa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-21 07:53:30.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/17/check/dsvm-tempest-cisco-zonemanager-job/edf215e/ : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":17},{"id":"d56ab3b7748cc4b9ad1bde7ab9ccc9e067fa4090","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-21 08:02:49.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/17/iscsi/ : SUCCESS in 1h 07m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/17/smb/ : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":17},{"id":"eff976d47b536303a51f1ead33b4b5cd685b20d7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-21 08:23:29.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/17/check-not-vote/ext-cinder-zuul/316825e : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":17},{"id":"87e9d774fda41aa01da40097f3e0f091137edf37","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-21 08:27:06.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"0d4a1c4df42119f5036e9476bbb3fcbe4a67644b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-21 08:27:49.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"99ff5af5972fd0de406f98cb32c25062ac22c2e9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-21 09:05:51.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/17/check/DellEMC_PowerFlex/96b3ee1/DellEMC_PowerFlex/263 : SUCCESS in 2h 32m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"01593b22abf86da03e9f7a9ee813887d97dbf4c0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-21 09:21:25.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/17/check-not-vote/ext-cinder-zuul/c191f02 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":17},{"id":"ac9c924d12f19fab86ba58c8a7d867c73609bcee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:24:28.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c4730bc24b646788de51d9b6c511b54 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e5ad02105d4fd58500e0aab7de7b2a : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e563c353942482ab18a57841b2fb9a0 : FAILURE in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/188d0fa0117946c88ccdac41b9a8c897 : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6754f49f39c8451b95043821f312b190 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95c8ce6bb91a43dda05058299e2e86c8 : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/ca2570539235493ea172dcf3f59e6e19 : SUCCESS in 1h 00m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c945fdd34ca34be4917f99487d17afb7 : SUCCESS in 1h 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68e585a9803c4815b8b7f77b9feb26ec : SUCCESS in 7m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e9343b193414dd690b9f623f10d9889 : SUCCESS in 12m 32s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87006581f1ab42ea9bd9be6616ecb34d : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac3652095915432c87b2a03ffb43f902 : SUCCESS in 6m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/770cc828ee3740d5be7316ae6bba8bf3 : SUCCESS in 7m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/115307c46a0541748f04fb980c39a110 : SUCCESS in 49m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5936e2ff49b7477d8e3112471cdc4886 : FAILURE in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a2aa2a4ea2f4cad935fccca43a77f04 : FAILURE in 27m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5c6690b6733f4b658ee680dc8920f986 : TIMED_OUT in 2h 09m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc89471614a84db08f1ec93d9de03c15 : SUCCESS in 1h 32m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/953123fa8d64469293586735cc107423 : SUCCESS in 1h 10m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b2af7ee143c4d2cb1c28f11b0055984 : SUCCESS in 1h 29m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e93defd56089469781d7485fad6023dc : SUCCESS in 1h 13m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0877d1816164a7f9a5abd7ef04fd7e4 : SUCCESS in 2h 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37c4245217c649259b564cc189cd9da8 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":17},{"id":"11765b2b372cf70e157e2eca17c11d27d7cbb303","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-21 09:26:45.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"72911897f93c712cb021e529244728f2d7e2dda3","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-21 09:27:39.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4d4ce3e647e682bae235dacf7cab1dd182d12007","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-21 09:52:05.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/90/767290/17/check/dsvm-tempest-cisco-zonemanager-job/5753e11/ : SUCCESS in 1h 19m 52s","accounts_in_message":[],"_revision_number":17},{"id":"af2a53df5d8872c8392a35492af805bc88872190","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-21 09:53:19.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767290/17/iscsi/ : SUCCESS in 1h 25m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/767290/17/smb/ : SUCCESS in 1h 14m 21s","accounts_in_message":[],"_revision_number":17},{"id":"574f3777e97eed2302e396a33cabc1f0ef8a9dcb","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-21 09:58:28.000000000","message":"Patch Set 17:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-17/Seagate-iSCSI : SUCCESS in 46:36.60\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-17/Seagate-FC : SUCCESS in 39:59.33","accounts_in_message":[],"_revision_number":17},{"id":"49af2816ab2dd10cdb1f5299245c42eb624ba4b5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-21 11:48:32.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767290/767290/17/check-not-vote/ext-cinder-zuul/56a801b : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":17},{"id":"4cee52aa5658bb8836fe041009985c88a4638a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 12:37:43.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4d9dc3764db451ca70ce2d6ab0ff2ef : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad4449b829343fb99636df44ad610e3 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aca653ce1d046159cfe99d28bc1a935 : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6528dfba2b484c0d9496559d59673b38 : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/455e50e96a9141e2af7ac651dca2a22e : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d82c1bb1864159b8534c349cccf21b : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/2b1338b707634f338c771675f7130b2b : SUCCESS in 59m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a23115220ef64651bedae2cd304f95eb : SUCCESS in 1h 19m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8ff58bf5fc945c0a5f96f1eb27b3f6e : SUCCESS in 6m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ab9f9c4e0bd1485cb482ffdb896cf603 : SUCCESS in 13m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79f9a295e10d4819aeee3793a7f96863 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea4f2222a1df47c0a81b7f349b983f43 : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a062845fe6184809a8b46fdb932f6796 : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/efd687c60e61404d8dd69e1b1a6375e9 : SUCCESS in 52m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dcd0ac39efeb41fabf158d75e5bc65ea : FAILURE in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/748f571c0a0842639a08745368500e34 : SUCCESS in 1h 27m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/201a8dbdb4c84094a3676b8ab79ada5f : FAILURE in 1h 18m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/22c3ad5e04db4fd6bf0072a060e6f686 : SUCCESS in 1h 31m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/19a681a0e3f746b49e5c4060e0c5c5d2 : SUCCESS in 1h 09m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6116a688e24f4ae48a9b479fe814ee39 : SUCCESS in 1h 52m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc8816436e0544a0ad765b850cf2734e : SUCCESS in 1h 17m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/938d436ed2b84d12b334b0ba9ea50920 : SUCCESS in 2h 05m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa70a69e0b8941ea9557d74140e5f812 : SUCCESS in 1h 03m 07s","accounts_in_message":[],"_revision_number":17},{"id":"49db55cc55b2bbcb87cce217444139271923cc64","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-21 14:13:23.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-iSCSI/a9f0bc4/ : SUCCESS in 40m 33s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-iSCSI-IPv6/f4eb3fb/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-NFS/1f64273/ : SUCCESS in 38m 50s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-NFS-IPv6/f0c8576/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/90/767290/17/upstream-check/cinder-cDOT-FCP/b1d972a/ : SUCCESS in 1h 35m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":17},{"id":"3e2a9938a895b2a4fdeffc01f0f3eaad6155fefa","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-21 14:13:28.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/90/767290/17/check/DellEMC_PowerStore_iSCSI/8bf6a81/DellEMC_PowerStore_iSCSI/312 : FAILURE in 50m 41s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/90/767290/17/check/DellEMC_PowerStore_FC/f589f84/DellEMC_PowerStore_FC/306 : FAILURE in 50m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":17},{"id":"3837b33a53637b60609788076aa25b59cd31208e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-21 14:16:48.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/90/767290/17/check/DellEMC_SC_FC/4a3e05e/DellEMC_SC_FC/252 : FAILURE in 2h 21m 04s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/90/767290/17/check/DellEMC_SC_iSCSI/3ed82e8/DellEMC_SC_iSCSI/252 : FAILURE in 1h 40m 02s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":17},{"id":"efe318fe7a439d6df76ebbe0e16e71429cd7623c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-21 14:21:25.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/90/767290/17/upstream-solidfire-check/cinder-solidfire-iSCSI/8fae54f/ : FAILURE in 1h 42m 42s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":17},{"id":"6a072907424f8727b7c6f6de84ca0320f579151e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-21 15:13:59.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/90/767290/17/check/DellEMC_PowerFlex/dc20832/DellEMC_PowerFlex/267 : FAILURE in 2h 35m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"e6361d960a40d7829acb9213776e1e7d9a708125","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-01-21 18:03:19.000000000","message":"Patch Set 17:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/767290-17/Lenovo-iSCSI : SUCCESS in 49:37.81\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/767290-17/Lenovo-FC : SUCCESS in 51:31.19","accounts_in_message":[],"_revision_number":17},{"id":"50d42973cc2dd58574c2b30605b0a54aaf78af37","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-21 18:21:28.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/90/767290/17/check/DellEMC_Unity_FC/a3b19d2/DellEMC_Unity_FC/504 : SUCCESS in 1h 37m 20s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/90/767290/17/check/DellEMC_Unity_iSCSI/26a37f3/DellEMC_Unity_iSCSI/468 : SUCCESS in 2h 00m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"fe9bb3c31d2684a644515ad89d56a7d38bdb5185","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-21 19:54:30.000000000","message":"Patch Set 17:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767290-17/Seagate-iSCSI : SUCCESS in 45:18.80\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767290-17/Seagate-FC : SUCCESS in 39:11.41","accounts_in_message":[],"_revision_number":17},{"id":"4a84d2f2868e488dffbe8ee30f216c3866c3f987","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-21 19:59:11.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/90/767290/17/check/DellEMC_VNX_iSCSI/28e5e2b/DellEMC_VNX_iSCSI/300 : SUCCESS in 2h 56m 00s\n- DellEMC_VNX_FC http://publiclogs.emc.com/90/767290/17/check/DellEMC_VNX_FC/0471012/DellEMC_VNX_FC/301 : SUCCESS in 1h 10m 29s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":17},{"id":"d73879de88a44942ff62927fffae5d639d75a41f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-21 20:26:31.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/17/20210122030036/NEC-ISCSI : SUCCESS in 2h 14m 40s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/767290/17/20210122030036/NEC-FC : SUCCESS in 2h 25m 53s","accounts_in_message":[],"_revision_number":17},{"id":"15b7e5ead08efe688ca359e8fb76998dfd7cb7a1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-21 20:54:15.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/90/767290/17/check/DellEMC_XtremIO_FC/61f2088/DellEMC_XtremIO_FC/584 : SUCCESS in 2h 08m 08s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/90/767290/17/check/DellEMC_XtremIO_iSCSI/703aa85/DellEMC_XtremIO_iSCSI/576 : SUCCESS in 1h 41m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":17},{"id":"03fc7416144b1a140f25c11554a0a9eebbfc1b3b","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-01-21 21:36:49.000000000","message":"Patch Set 17:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/767290-17/HPMSA-iSCSI : FAILURE in 1:05:49\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/767290-17/HPMSA-FC : SUCCESS in 40:54.20","accounts_in_message":[],"_revision_number":17},{"id":"7058f6f2fca09c3179aaecf3664bfb7da953f328","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-21 22:28:12.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)\n\nSee comment inline about a followup.\n\nThird-party CI is passing and TOYOU team is committed to getting it to respond on all patches before M-3.  Looks like most of the concerns have been addressed.  Only a +1 right now to give Jay a chance to make sure his concerns have been met. \n\nFor the purposes of the new driver merge deadline for Wallaby, this patch will be fine as long as it\u0027s approved before 22:00 UTC tomorrow (Friday 22 January).","accounts_in_message":[],"_revision_number":17},{"id":"7abe20e829c823712edf6c139f7d378411b5a691","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-22 02:05:49.000000000","message":"Patch Set 17:\n\n\u003e Patch Set 17: Code-Review+1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e See comment inline about a followup.\n\u003e \n\u003e Third-party CI is passing and TOYOU team is committed to getting it to respond on all patches before M-3.  Looks like most of the concerns have been addressed.  Only a +1 right now to give Jay a chance to make sure his concerns have been met. \n\u003e \n\u003e For the purposes of the new driver merge deadline for Wallaby, this patch will be fine as long as it\u0027s approved before 22:00 UTC tomorrow (Friday 22 January).\n\nThanks for your reviewing. I have already fixed the get_driver_options issue and will adress it together with Sean\u0027s commentin a future separate patch.","accounts_in_message":[],"_revision_number":17},{"id":"db06a293eef13cf094b9c0c4615f9ffb5e0ef346","author":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"date":"2021-01-22 02:17:05.000000000","message":"Patch Set 17:\n\n(1 comment)\n\nreply to the get_driver_options method.","accounts_in_message":[],"_revision_number":17},{"id":"5047398117a6302c7f0ee5a9b8c0d94260507ef6","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-01-22 17:04:45.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\nIt appears that the major comments have been addressed.  The CI looks pretty good to me.  I think that this can merge now.  Thank you for the attention to all the reviews.","accounts_in_message":[],"_revision_number":17},{"id":"55d689b52da9c1c55d69cfb1fd73e855b2f15b22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:13:02.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"fd50066cc87128d2847c1506e0a9116de5ddc7eb","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-01-22 17:17:21.000000000","message":"Patch Set 17:\n\n\u003e Patch Set 17: Code-Review+2 Workflow+1\n\u003e \n\u003e It appears that the major comments have been addressed.  The CI looks pretty good to me.  I think that this can merge now.  Thank you for the attention to all the reviews.\n\nGreat! Thank for your reviewing again and again. :-)","accounts_in_message":[],"_revision_number":17},{"id":"5209741e0a1bba6af3a18c497691057002614912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:42:26.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b7d320d4edf4427b1fad7cdaf840a61 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb75938386e4e7fadbbf64ab5552f5d : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d97e10de3af9440c9c81b8d3f73d7b24 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf30df5403834a0bb5bb8b83ac215ffe : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872259fec5f24629b11d4b3d90cc5b45 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/4d81a7b3cb614b50896141f92aeabc0f : SUCCESS in 1h 03m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8f88477cf7b4dd68fb664e395d5986e : SUCCESS in 1h 15m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fccf7721b59443d58f7afc24418808be : SUCCESS in 6m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a4105ca133a34f2899391800a72f5935 : SUCCESS in 1h 10m 55s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/000ebe9af3194ae58aaebb700a339d85 : SUCCESS in 1h 13m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06d792c631bf48459749a356018e33a3 : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":17},{"id":"ef598bf631f3d19adc8edcefd76eb72e2e150ff3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:42:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"2cc840fa18490b8a63c56882ecec71f65149fbd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:48:40.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709ae5e0511b4494a646581bf626f178 : SUCCESS in 4m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26bd4551161b4bfabb6b374b6f55d6de : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"fe0822b1bcd42d3d77e06b3e383d5ee6b22f25e9","revisions":{"4ab0df46ea2a318c286057585fdc1ab0f7fd9ec6":{"kind":"REWORK","_number":1,"created":"2020-12-16 09:48:03.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/1"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ab0df46ea2a318c286057585fdc1ab0f7fd9ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ab0df46ea2a318c286057585fdc1ab0f7fd9ec6"}]},"branch":"refs/heads/master"},"c6349b295de91a3cc7450dcb81b9f3cdf78818d7":{"kind":"REWORK","_number":2,"created":"2020-12-20 09:48:52.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/2"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-20 09:40:27.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6349b295de91a3cc7450dcb81b9f3cdf78818d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6349b295de91a3cc7450dcb81b9f3cdf78818d7"}]},"branch":"refs/heads/master","description":"format driver code,use command \u0027tox -e pep8\u0027"},"6243f755b2b7c118b32852a928342deae6c20c49":{"kind":"REWORK","_number":3,"created":"2020-12-21 02:32:08.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/3"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-21 02:26:07.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6243f755b2b7c118b32852a928342deae6c20c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6243f755b2b7c118b32852a928342deae6c20c49"}]},"branch":"refs/heads/master","description":"use command \u0027tox -e genopts\u0027,add acs5000 driver into cinder/opts.py"},"08918ef6626c623fa8384a0a0fb655a242d6a5fc":{"kind":"REWORK","_number":4,"created":"2020-12-29 09:07:18.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/4"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-29 09:01:11.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08918ef6626c623fa8384a0a0fb655a242d6a5fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08918ef6626c623fa8384a0a0fb655a242d6a5fc"}]},"branch":"refs/heads/master"},"620ddfe356284ceb6bb9b391b11e700d09fb6e5f":{"kind":"REWORK","_number":5,"created":"2021-01-07 10:24:07.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/5"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-07 10:17:52.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration (host assisted)\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/620ddfe356284ceb6bb9b391b11e700d09fb6e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/620ddfe356284ceb6bb9b391b11e700d09fb6e5f"}]},"branch":"refs/heads/master"},"cf0523f94d82c18e6cd740f42f9b8794e4dd3a5e":{"kind":"REWORK","_number":6,"created":"2021-01-11 09:52:00.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/6"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-11 09:43:45.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration (host assisted)\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf0523f94d82c18e6cd740f42f9b8794e4dd3a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf0523f94d82c18e6cd740f42f9b8794e4dd3a5e"}]},"branch":"refs/heads/master"},"10b30d97a850507637964c29eefd6910a7db4efd":{"kind":"REWORK","_number":7,"created":"2021-01-12 03:04:36.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/7"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-12 02:57:53.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration (host assisted)\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10b30d97a850507637964c29eefd6910a7db4efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10b30d97a850507637964c29eefd6910a7db4efd"}]},"branch":"refs/heads/master"},"698b6d3cb4bde63e1dd5eeafa983a9f7334bfdd5":{"kind":"REWORK","_number":8,"created":"2021-01-14 01:58:44.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/8"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-14 01:52:42.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration (host assisted)\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/698b6d3cb4bde63e1dd5eeafa983a9f7334bfdd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/698b6d3cb4bde63e1dd5eeafa983a9f7334bfdd5"}]},"branch":"refs/heads/master"},"2555c77c0e83a9815f7634d831e34dc059f20039":{"kind":"REWORK","_number":9,"created":"2021-01-14 02:32:41.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/9"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-14 02:26:28.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration (host assisted)\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2555c77c0e83a9815f7634d831e34dc059f20039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2555c77c0e83a9815f7634d831e34dc059f20039"}]},"branch":"refs/heads/master"},"e85567e3227487106dcb92beeb7ccdc0950b9bf3":{"kind":"REWORK","_number":10,"created":"2021-01-14 10:18:07.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/10"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-14 10:06:43.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration (host assisted)\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e85567e3227487106dcb92beeb7ccdc0950b9bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e85567e3227487106dcb92beeb7ccdc0950b9bf3"}]},"branch":"refs/heads/master"},"725e1118b0fda5baecfd62153dd52c62c53596cd":{"kind":"REWORK","_number":11,"created":"2021-01-18 18:00:22.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/11"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-18 17:51:02.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration (host assisted)\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/725e1118b0fda5baecfd62153dd52c62c53596cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/725e1118b0fda5baecfd62153dd52c62c53596cd"}]},"branch":"refs/heads/master"},"ba5ad54548d700a07ebbb62368d542c2fa9a6aea":{"kind":"REWORK","_number":12,"created":"2021-01-19 03:49:50.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/12"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-19 03:18:38.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba5ad54548d700a07ebbb62368d542c2fa9a6aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba5ad54548d700a07ebbb62368d542c2fa9a6aea"}]},"branch":"refs/heads/master"},"79adcaa4fcdf22e725032032ae859af9e00c35fc":{"kind":"REWORK","_number":13,"created":"2021-01-19 10:51:32.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/13"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-19 10:50:32.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79adcaa4fcdf22e725032032ae859af9e00c35fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79adcaa4fcdf22e725032032ae859af9e00c35fc"}]},"branch":"refs/heads/master"},"033b8f5d546b1b3387ad36890c4f44f559ebf86b":{"kind":"REWORK","_number":14,"created":"2021-01-20 09:43:04.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/14"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-20 09:42:22.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/033b8f5d546b1b3387ad36890c4f44f559ebf86b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/033b8f5d546b1b3387ad36890c4f44f559ebf86b"}]},"branch":"refs/heads/master"},"06690be3238964eb6c6fe4dcb94cae2511abcdd9":{"kind":"REWORK","_number":15,"created":"2021-01-20 10:03:50.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/15"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-20 10:03:31.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06690be3238964eb6c6fe4dcb94cae2511abcdd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06690be3238964eb6c6fe4dcb94cae2511abcdd9"}]},"branch":"refs/heads/master"},"430242c354c5ab31025feef6ba285fbe9967fc78":{"kind":"REWORK","_number":16,"created":"2021-01-20 10:18:21.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/16"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-20 10:17:59.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/430242c354c5ab31025feef6ba285fbe9967fc78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/430242c354c5ab31025feef6ba285fbe9967fc78"}]},"branch":"refs/heads/master"},"fe0822b1bcd42d3d77e06b3e383d5ee6b22f25e9":{"kind":"REWORK","_number":17,"created":"2021-01-20 18:29:46.000000000","uploader":{"_account_id":32833,"name":"yangheng","email":"yangheng@toyou.com.cn","username":"yangheng"},"ref":"refs/changes/90/767290/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/767290/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/767290/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/767290/17"}}},"commit":{"parents":[{"commit":"108c554dc3a853b053fbc1360f27ef13a0dc8335","subject":"Merge \"Add CHAP support to Dell EMC PowerStore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/108c554dc3a853b053fbc1360f27ef13a0dc8335"}]}],"author":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2020-12-16 09:42:56.000000000","tz":0},"committer":{"name":"yangheng","email":"yangheng@toyou.com.cn","date":"2021-01-20 18:28:51.000000000","tz":0},"subject":"Add Cinder driver for TOYOU ACS5000","message":"Add Cinder driver for TOYOU ACS5000\n\n* Supported Protocol\n - iSCSI\n\n* Supported Feature\n - Volume Create/Delete\n - Volume Attach/Detach\n - Snapshot Create/Delete\n - Create Volume from Snapshot\n - Get Volume Stats\n - Copy Image to Volume\n - Copy Volume to Image\n - Clone Volume\n - Extend Volume\n - Volume Migration\n\nThirdPartySystems: TOYOU ACS5000 CI\nChange-Id: Ia7d2056e0a49032654812da74dac514dfac83529\nImplement: blueprint toyou-acs5000-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0822b1bcd42d3d77e06b3e383d5ee6b22f25e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0822b1bcd42d3d77e06b3e383d5ee6b22f25e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
