)]}'
{"id":"openstack%2Fcinder~767334","triplet_id":"openstack%2Fcinder~stable%2Fussuri~I84629cde6d27d6889a30d7c13d99f57c660adde7","project":"openstack/cinder","branch":"stable/ussuri","topic":"bug/1890254-stable/ussuri","hashtags":[],"change_id":"I84629cde6d27d6889a30d7c13d99f57c660adde7","subject":"[SVF]:Fix clone fcmap not being deleted in cleanup","status":"MERGED","created":"2020-12-17 06:41:07.000000000","updated":"2021-03-05 17:59:35.000000000","submitted":"2021-03-05 17:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"767334-bug/1890254-stable/ussuri","cherry_pick_of_change":744664,"cherry_pick_of_patch_set":14,"meta_rev_id":"a4c013d1ba94c62de784d440e5e86fd14aa9e7cb","_number":767334,"virtual_id_number":767334,"owner":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 17:54:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-05 15:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":2,"date":"2021-03-04 13:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-05 15:16:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"_account_id":23802,"name":"cao biao","email":"bjcb@cn.ibm.com","username":"caobiao"}],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23802,"name":"cao biao","email":"bjcb@cn.ibm.com","username":"caobiao"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 06:44:47.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"state":"REVIEWER"},{"updated":"2020-12-17 06:48:42.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2020-12-17 06:48:42.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":23802,"name":"cao biao","email":"bjcb@cn.ibm.com","username":"caobiao"},"state":"REVIEWER"},{"updated":"2020-12-17 06:50:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 07:07:22.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-17 07:42:42.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-17 08:18:47.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 08:52:18.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 09:05:12.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 09:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 10:01:55.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 20:36:30.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-10 14:39:00.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2021-02-24 06:41:57.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-03-03 05:44:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-03-03 06:08:48.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-03 08:25:29.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-03 08:26:23.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 13:46:11.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-05 15:16:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"4b80b3d39975cfd87098b5d5665b05014d6e4494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-17 06:41:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"df14e1d595295fefdad686d5c84eb9d0de3d9322","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 06:50:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/34/767334/1/check/3par-iscsi-driver-primera/af87ffa/af87ffa.html : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"21511e332053dead9617089e0dc22c8c8ed3f064","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2020-12-17 06:51:28.000000000","message":"Patch Set 1: Code-Review+1\n\nthanks :)","accounts_in_message":[],"_revision_number":1},{"id":"0f4c9990d824e2ac7a1fa83aca7256181c3b067a","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 06:52:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/34/767334/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/686fb75/686fb75.html : FAILURE in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fbef5a01f39312e38a9ea43195cab74dd242aad6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-17 07:07:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/34/767334/1/check/dsvm-tempest-cisco-zonemanager-job/9adf013/ : FAILURE in 21m 10s","accounts_in_message":[],"_revision_number":1},{"id":"50d159380be4b041ce67e79bf6aba3ff8969284c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-17 07:42:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767334/1/iscsi/ : SUCCESS in 1h 00m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/767334/1/smb/ : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d1879ad5f3efda215bf972b620e1c0932f59556f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-17 08:18:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/34/767334/1/ : RETRY_LIMIT in 2m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4b4981ccfb3c1834b115c25cc07fc863863a0a95","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-17 08:52:18.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/34/767334/1/check/tempest-dsvm-full-bionic/0e7cfb6/ : SUCCESS in 1h 55m 43s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/34/767334/1/check/tempest-dsvm-full-bionic-py3/267308a/ : SUCCESS in 1h 43m 22s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/34/767334/1/check/devstack-plugin-ceph-dsvm-bionic-py3/b5ff93b/ : SUCCESS in 2h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"227554f016d45a25b659557fa229a6c2ead4bd60","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-17 09:05:12.000000000","message":"Patch Set 1:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/34/767334/1/check/dsvm-tempest-zadara-driver/e970dde : FAILURE in 1h 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9aed526f1c43b84c34a8c5b82653c5dcb2c144bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 09:45:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67267955e5b84173987258c27af65650 : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aabc930abbb43219b3305d50f6ee2cd : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63ff13092fc640979d6f8648c93b3c99 : FAILURE in 8m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8543b55bee745f5873853a68c7f4caf : FAILURE in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7803b8a0df4e4729bba8266ffc0d2790 : SUCCESS in 15m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/93ea6d6c26d040eab71c871f4fc248b6 : SUCCESS in 1h 12m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c4627fa92504809807926f737c9b808 : SUCCESS in 1h 17m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35e06e2923ff4cf79f00f8febe96b228 : SUCCESS in 7m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c66c2b0328654942a886e6eb9f345122 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/440b9ccce2e1402daca18ced52acc9a5 : SUCCESS in 7m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6cc9b1ba598849c88c9cd03efe20d529 : SUCCESS in 7m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e0e77b1da3044de987428299b6e860bf : SUCCESS in 36m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dbcf75a8e85343dfb74b32c644116095 : SUCCESS in 7m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ecd70e2ac6424dc8bfa03228d76341cc : SUCCESS in 1h 18m 33s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a091c2bd3b974ef6ae9655a1f1ce9036 : FAILURE in 1h 27m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8ab9274aca2b4e4693765a0927b1dfcc : SUCCESS in 1h 21m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2eff44c9c42948d48fb53edbca815e15 : SUCCESS in 1h 26m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c3a44cf95004b1182fc45151b4cc985 : FAILURE in 1h 09m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a255bd64346467a9a57342506880f4b : SUCCESS in 2h 03m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/650366a3e5d24f97bf6fa0979e2ae057 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":1},{"id":"750e78c7b8081580ba5f74a6e016cc4d0655b30f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-17 10:01:55.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-34-767334-1/ns5_nfs : SUCCESS in 3h 8m 49s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-34-767334-1/ns5_iscsi : SUCCESS in 2h 59m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8fe628ebeb9e3b93bcba81c9efb32f71a9ab7b78","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-17 20:36:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/767334/1/20201218032720/NEC-FC : SUCCESS in 2h 02m 52s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/767334/1/20201218032720/NEC-ISCSI : SUCCESS in 2h 09m 08s","accounts_in_message":[],"_revision_number":1},{"id":"dedeb84edaee72a6773aee556b354fc3ce1d7afd","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-10 14:39:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"326be2d0571dfd76a60ad99fe2c67ca2392e0329","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-10 14:39:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d54c52aba88228f9311af79991112ed94160cc5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-10 15:40:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/34/767334/1/check/dsvm-tempest-cisco-zonemanager-job/3c1a7e5/ : FAILURE in 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c693157315908c779fb321812dc205824c0aca85","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-10 15:42:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767334/1/iscsi/ : SUCCESS in 1h 02m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/767334/1/smb/ : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":1},{"id":"581c6e5c550c6be774ea6a4d5c91fbdfd2bc8f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:46:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c345714664d144cfa0d18f5d8ba21de3 : FAILURE in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff9407629dff446aabe6d1a6fcc9ce94 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec3feab031e4bc0b5204b5e090d6c5a : FAILURE in 8m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f1e1383cc7e4698b9d54340eeabd056 : FAILURE in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e53cc68f064ccf96bdc6e8fdc8d448 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/3f1ce62628b444bca2a72b05ac721952 : SUCCESS in 1h 08m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57972e14e48a4a6592da6ed630cee9a7 : SUCCESS in 1h 27m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/304eec3340d2438fa0b89bf55c01f3bc : SUCCESS in 6m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1337a1702fb64265a121aaf66ffb937d : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b028ca4f9c1b48ddb297dd4fb3950078 : SUCCESS in 6m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/562df8f31c2a4cb486c7bf29694a83ba : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/79f79aff94744d979a82d51ceaa8cead : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a9295b0931c4e1e8affdd756afa26cb : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8d7eb9c946874c428f3d6877a44984d9 : SUCCESS in 1h 07m 06s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c6b8ba6ca106454b9483bd6f9caddf47 : FAILURE in 1h 32m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/61cac5425d8d4dcb8e37d2c12e11d1f2 : SUCCESS in 1h 32m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/498dc98b3cf8449285f404f088e4f701 : SUCCESS in 1h 15m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/03f3310b719a4d6c9cd1eb776aa86879 : FAILURE in 1h 07m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a1b6c7dbbe049e38dd28a8f97428be3 : SUCCESS in 2h 02m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cc33c2844ea40ff97e19ef8037fc6e2 : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b2a07899e26a5c12f84e92ea5e098a2bcb4b08c3","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2021-02-10 17:47:09.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e recheck\n\nZuul failed because of below reason,\nddt was used for the failed test cases, and this statement \"@ddt.ddt\" is required before the class of the failed test cases belongs to. \"@ddt.ddt\" has pushed to git from code review \"[SVF]:Reduce slowness by caching pool information\", here is the code change with which test cases are failing https://review.opendev.org/c/openstack/cinder/+/767478/4/cinder/tests/unit/volume/drivers/ibm/test_storwize_svc.py#8146\n\nWaiting for \"https://review.opendev.org/c/openstack/cinder/+/767478\" to be merged, once \"[SVF]:Reduce slowness by caching pool information\" code is merged Zuul will pass for this patchset.","accounts_in_message":[],"_revision_number":1},{"id":"65d9b68197cd9108cfaef7c86c5051a9e912faa0","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-24 06:41:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\n[1] needs to merge for zuul to pass here\n\n[1] https://review.opendev.org/c/openstack/cinder/+/767478","accounts_in_message":[],"_revision_number":1},{"id":"e6cda8535a25b7834662a7505b8d9418c11d1df3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2021-03-03 05:35:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"42b50863a819a6fd2220d9cc18279eed043bf7c6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-03 05:36:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0dadd24a6c553d7c0cab70e8889bbf15d042b517","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-03 05:37:14.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"99afdb74c34e2031bc539f2a51ee10f6280e3b9d","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2021-03-03 05:39:36.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nHi Rajat,\nAddressed review comment.","accounts_in_message":[],"_revision_number":2},{"id":"3fce9a196dd0dc15d125b4d579e8f0c0fadd9713","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-03 05:44:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/34/767334/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/437ac8e/ : FAILURE in 7m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/34/767334/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4226e53/ : FAILURE in 6m 47s","accounts_in_message":[],"_revision_number":2},{"id":"33b9c5946e8898d9bbb87faae9a30df616d408d4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-03 06:08:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/34/767334/2/check/dsvm-tempest-cisco-zonemanager-job/f3bb42f/ : FAILURE in 25m 56s","accounts_in_message":[],"_revision_number":2},{"id":"efbc39e7a7f0d178710bdd86711669778e74222c","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-03 06:08:48.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/34/767334/2/check/dsvm-tempest-TOYOU-ACS5000-driver/f26c3ab : FAILURE in 30m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0159b8335b1bd66dc7af2048d53c851b39572859","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-03 06:11:22.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-34-767334-2/ns5_nfs : FAILURE in 34m 14s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-34-767334-2/ns5_iscsi : FAILURE in 34m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"c3b54efe06e24131be90fb3d399f6d4d0c62ab2b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-03 06:25:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/34/767334/2/check/iscsi-driver-primera/9e392c8/9e392c8.html : FAILURE in 47m 50s","accounts_in_message":[],"_revision_number":2},{"id":"bbef87f4692ba30af8e7cf0ac77f3225eb88ec52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 07:57:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/432976f4ab1245de8c39e4e9e0d6679b : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86acfada056f4d0484647569b6894207 : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/601321023d9345d796ff10d993a38720 : SUCCESS in 8m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68b82f12897441ebb81d5f39f91960c9 : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cf0295fe424b109ff1c4e6f984fb04 : SUCCESS in 14m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/243fc81bba1844c3b44419b279205b40 : SUCCESS in 1h 21m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/298ce063d9c5470488decdf87648b71c : SUCCESS in 1h 17m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8492c2f0beea4f74a4e1c1d21128bc13 : SUCCESS in 7m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d2af7d51f3144599bf88caa42e2b156b : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a309165f542047089f095a14ac029a1d : SUCCESS in 7m 27s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0c168fc620684623b27b2715ac84e472 : SUCCESS in 7m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/11a03ba184e8482387c2f440b9e54378 : SUCCESS in 45m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/82d13e0a6ae64345bf74cb4e0c17a3f3 : SUCCESS in 7m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5599be91d00a4f9883ceb892e4b8e389 : SUCCESS in 1h 32m 29s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fe74b10b88dd4fa58f84d7a2c8c99500 : SUCCESS in 1h 46m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dd452eebab9d4e92a684574cb6832042 : SUCCESS in 1h 32m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f7b88826b57d47f38a47826b2fda23a8 : SUCCESS in 1h 36m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b9542701e5fb453181554f5e3ac76dc4 : FAILURE in 1h 16m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/704f73e1b2ba430fb5286d3d43a7bcdb : SUCCESS in 2h 16m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13539a934fa5474086d382c40cb277dc : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":2},{"id":"197418dfea5c9ebb63987d23b5c8ce5cd3419f7c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-03 08:25:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/34/767334/2/upstream-solidfire-check/cinder-solidfire-iSCSI/a54e540/ : FAILURE in 26m 50s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"e7525ea246b6c62fec7d47ac20280c726bddcc1a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-03 08:26:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/34/767334/2/upstream-check/cinder-cDOT-iSCSI/c4a99ec/ : FAILURE in 1m 41s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/34/767334/2/upstream-check/cinder-cDOT-iSCSI-IPv6/d74acca/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/34/767334/2/upstream-check/cinder-cDOT-NFS/4e2eb6e/ : FAILURE in 19m 39s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/34/767334/2/upstream-check/cinder-cDOT-NFS-IPv6/b552ad3/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/34/767334/2/upstream-check/cinder-cDOT-FCP/f8e1a22/ : FAILURE in 28m 07s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"6391615c635370d86102393e9d98f80e0b95b570","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-03 08:30:34.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/34/767334/2/check/tempest-dsvm-full-bionic/ba8a2f0/ : SUCCESS in 1h 52m 17s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/34/767334/2/check/tempest-dsvm-full-bionic-py3/d20a6a7/ : SUCCESS in 1h 57m 37s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/34/767334/2/check/devstack-plugin-ceph-dsvm-bionic-py3/90351d0/ : TIMED_OUT in 2h 47m 44s","accounts_in_message":[],"_revision_number":2},{"id":"2944d3281a592683db4e5f58d4fb0944021da81e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-03-03 13:35:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/34/767334/2/ : RETRY_LIMIT in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"38cb35b8a04f65164f32a8c454b41afcc1b1a139","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-03 23:11:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/34/767334/2/20210304055557/NEC-ISCSI : SUCCESS in 2h 13m 26s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/34/767334/2/20210304055557/NEC-FC : SUCCESS in 2h 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"73aecae6b71d77fcf5f73fe1e8b9c307781565c7","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-04 13:46:11.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks like a good backport.  Thank you.","accounts_in_message":[],"_revision_number":2},{"id":"16676860ec378ecbc068c95a5ec1893f707c2d82","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-05 15:16:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nAgree with Jay that this is worth backporting.","accounts_in_message":[],"_revision_number":2},{"id":"1731e7866392b5ff6d0db8ce997eb74136328be8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 15:26:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"add0818d1010a529933574961d2d7c3f00e501d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:54:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ee4cd3eb4414a269b8f2b91b5a53947 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3eeb34c81c40febd116dfb20dfcdf2 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90b34131475a41c281a8bc7a6cb3ef1d : SUCCESS in 13m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/536494a6075543b78d90b467e4a1949f : SUCCESS in 15m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/625bef1506ed4bbc847414e482a98f56 : SUCCESS in 19m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/0ee8fab7e6da4baba348413dc5f3db04 : SUCCESS in 1h 17m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/41b34238a87248299b039a99df4d6be0 : SUCCESS in 1h 15m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba2e15fd7fd24c43a966363ac7bc7764 : SUCCESS in 8m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad4dc288c8664707aedf405e44e078b4 : SUCCESS in 1h 28m 23s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a77de22848dc474782bfc7c67d15a7f6 : SUCCESS in 1h 12m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/923805c507524c91aed4768500081343 : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f93f986ff4c7b8311087bba19ec4044a9b7c7c75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:54:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a4c013d1ba94c62de784d440e5e86fd14aa9e7cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:59:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303184f0b144406793de82fd78815dc0 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78dfa61b8f8b40c0bc3047d015cc52bf : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c469c1294d5a09f926905414f505d6e84ebaf506","revisions":{"a3a322669dfb8e2fa9bca8b982ee4042016b31df":{"kind":"REWORK","_number":1,"created":"2020-12-17 06:41:07.000000000","uploader":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"ref":"refs/changes/34/767334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/767334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/767334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/767334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/767334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/767334/1"}}},"commit":{"parents":[{"commit":"2f0d236440b6816ae2e50ae10192777fcf4117dd","subject":"Merge \"Fix volume rekey during clone\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f0d236440b6816ae2e50ae10192777fcf4117dd"}]}],"author":{"name":"GirishChilukuri","email":"girish.chilukuri@ibm.com","date":"2020-10-06 19:09:34.000000000","tz":0},"committer":{"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","date":"2020-12-17 06:41:07.000000000","tz":0},"subject":"[SVF]:Fix clone fcmap not being deleted in cleanup","message":"[SVF]:Fix clone fcmap not being deleted in cleanup\n\n[Spectrum Virtualize Family] check_vdisk_fc_mappings is not deleting\nclone fc maps, if the flashcopy status is in copying autodelete is set\nto on and copy progress is 100%. when multiple clones and snapshots\nare created using common source volume, deleting the source volume\nis not removing all the fcmaps involved in mapping with source volume.\n\nThis patch fixes the issue by removing the clone fcmap and then\nremoving the snapshot fcmap when the source has more than one fcmap.\n\nCloses-Bug: #1890254\n\nChange-Id: I84629cde6d27d6889a30d7c13d99f57c660adde7\n(cherry picked from commit 3570a79a16127fd42216aa40b4f492e40533f0ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3a322669dfb8e2fa9bca8b982ee4042016b31df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3a322669dfb8e2fa9bca8b982ee4042016b31df"}]},"branch":"refs/heads/stable/ussuri"},"c469c1294d5a09f926905414f505d6e84ebaf506":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-03 05:35:50.000000000","uploader":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"ref":"refs/changes/34/767334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/767334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/767334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/767334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/767334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/767334/2"}}},"commit":{"parents":[{"commit":"2f0d236440b6816ae2e50ae10192777fcf4117dd","subject":"Merge \"Fix volume rekey during clone\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f0d236440b6816ae2e50ae10192777fcf4117dd"}]}],"author":{"name":"GirishChilukuri","email":"girish.chilukuri@ibm.com","date":"2020-10-06 19:09:34.000000000","tz":0},"committer":{"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","date":"2021-03-03 05:34:50.000000000","tz":0},"subject":"[SVF]:Fix clone fcmap not being deleted in cleanup","message":"[SVF]:Fix clone fcmap not being deleted in cleanup\n\n[Spectrum Virtualize Family] check_vdisk_fc_mappings is not deleting\nclone fc maps, if the flashcopy status is in copying autodelete is set\nto on and copy progress is 100%. when multiple clones and snapshots\nare created using common source volume, deleting the source volume\nis not removing all the fcmaps involved in mapping with source volume.\n\nThis patch fixes the issue by removing the clone fcmap and then\nremoving the snapshot fcmap when the source has more than one fcmap.\n\nCloses-Bug: #1890254\n\nChange-Id: I84629cde6d27d6889a30d7c13d99f57c660adde7\n(cherry picked from commit 3570a79a16127fd42216aa40b4f492e40533f0ed)\n(cherry picked from commit 6e592a8dd49a85f8658ffeeb8c95a9d39f1d8507)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c469c1294d5a09f926905414f505d6e84ebaf506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c469c1294d5a09f926905414f505d6e84ebaf506"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
