)]}'
{"id":"openstack%2Fcinder~767440","triplet_id":"openstack%2Fcinder~master~I97afdc561b346abb87c3801aaab4b9f3e7492d6d","project":"openstack/cinder","branch":"master","topic":"backup-dependents-fail","hashtags":[],"change_id":"I97afdc561b346abb87c3801aaab4b9f3e7492d6d","subject":"backup dependents failure","status":"NEW","created":"2020-12-17 01:07:37.000000000","updated":"2021-03-25 11:13:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7fa52e7e82b3075fd87175f91e35dbf2a779076","_number":767440,"virtual_id_number":767440,"owner":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-19 05:56:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"all":[{"value":1,"date":"2021-03-25 11:13:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 01:46:38.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 01:49:59.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 02:20:34.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 03:50:31.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-17 04:03:20.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 04:08:18.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-17 04:17:34.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 04:53:37.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 05:14:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 06:18:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-17 06:51:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 21:48:09.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 22:41:31.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-19 07:12:50.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2020-12-19 07:59:54.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2020-12-19 08:08:03.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2020-12-19 08:12:49.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2020-12-19 08:13:12.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2020-12-19 08:16:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-19 08:50:32.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-19 08:56:35.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-19 09:28:27.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2020-12-19 10:03:50.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2020-12-21 11:50:27.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-03-25 11:13:35.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"}],"messages":[{"id":"afbbacf234f5b421aef08d0fa0f8ebac5ada635e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"date":"2020-12-17 01:07:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"267a6fd926df6f452612c719b33ec826cc00461a","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 01:46:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/40/767440/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4185c09/4185c09.html : FAILURE in 37m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0374915bd2598bbdfb8004150b0339f7225393dc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-17 01:49:59.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767440/767440/1/check-not-vote/ext-cinder-zuul/a248fb5 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":1},{"id":"879adcdbaa4d449728b89091965557c302ef4d40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"date":"2020-12-17 02:11:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"392691f04cf998ef204d6b4d212c48e80e90095a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-17 02:20:34.000000000","message":"Patch Set 2:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/40/767440/2/check/dsvm-tempest-zadara-driver/0f149f6 : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79116aae03c8e75c7aee9eb305bf5ee2163b827e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"date":"2020-12-17 02:44:18.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0033d99c7d81fb1436acb091551813821c664bb8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-17 03:32:44.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767440/767440/3/check-not-vote/ext-cinder-zuul/13456c8 : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":3},{"id":"189fb4c1247e749b30cc2fc72efd72ab48a29b76","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-17 03:50:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767440/3/iscsi/ : SUCCESS in 1h 05m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/767440/3/smb/ : SUCCESS in 55m 20s","accounts_in_message":[],"_revision_number":3},{"id":"0d6664f6229c5e4942601c5f6a254b129c983927","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-17 04:03:20.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-1/ns5_nfs : SUCCESS in 2h 55m 6s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-1/ns5_iscsi : SUCCESS in 2h 51m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"68a9b9d4541998289d50150ed68cf2a04161b282","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-17 04:08:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/40/767440/3/check/dsvm-tempest-cisco-zonemanager-job/dafadf4/ : SUCCESS in 1h 18m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a6655b920fa9b552932bf8efea7a31fdefb443e8","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 04:09:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/40/767440/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d04b1c8/d04b1c8.html : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":3},{"id":"3aca0ab739173cad79d1d2ec8a02d7c63244de5c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-17 04:17:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/40/767440/3/ : SUCCESS in 1h 25m 46s","accounts_in_message":[],"_revision_number":3},{"id":"354f531dc6013381c7fb5ca4a35748e17515c386","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-17 04:53:37.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/40/767440/3/check/tempest-dsvm-full-focal-py3/9886af5/ : SUCCESS in 1h 45m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/40/767440/3/check/devstack-plugin-ceph-dsvm-focal-py3/656b0ec/ : SUCCESS in 2h 02m 14s","accounts_in_message":[],"_revision_number":3},{"id":"4e81be12131f41af1109867f163802b8755bed80","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-17 05:14:28.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/1/20201217100741/NEC-ISCSI : SUCCESS in 2h 00m 57s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/1/20201217100741/NEC-FC : FAILURE in 4h 06m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b882e8cd49403f29d350dfa41a43184bb94adb58","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-17 05:28:00.000000000","message":"Patch Set 3:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/40/767440/3/check/dsvm-tempest-zadara-driver/22015e3 : FAILURE in 1h 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"11759df3bb80c7ffdaba0791a303502d9001c4c0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-17 05:32:29.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-2/ns5_nfs : SUCCESS in 3h 20m 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-2/ns5_iscsi : SUCCESS in 3h 5m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"a5a7e405737c5daa8ff808420275700f32bf75a7","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 05:49:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/40/767440/3/check/3par-iscsi-driver-primera/910724d/910724d.html : TIMED_OUT in 3h 00m 51s","accounts_in_message":[],"_revision_number":3},{"id":"bbd04ec7534a92011aedea617fd18ca0ad41ac1c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-17 06:18:20.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/40/767440/3/check/DellEMC_Unity_FC/1bc68ae/DellEMC_Unity_FC/318 : SUCCESS in 1h 18m 00s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/40/767440/3/check/DellEMC_Unity_iSCSI/930c0d2/DellEMC_Unity_iSCSI/301 : SUCCESS in 1h 54m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"23aa5d441f0fc772de8ce54fcb3308571729f1c2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-17 06:22:55.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-3/ns5_nfs : SUCCESS in 3h 25m 32s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-3/ns5_iscsi : SUCCESS in 3h 10m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"db66b59d3f29c75b7a4370e2ce0a9db4c455e002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 06:51:36.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/334d0881bf0a4280b6f20137bb8a322a : FAILURE in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e78727f34d34cd8adc107bc3d839762 : FAILURE in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44e4686bc9074bbd87c55bb7d4e1492c : FAILURE in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7101855265f44e40be71d93b34108f7b : FAILURE in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c0fa030e85644ec81087399d6d2117c : FAILURE in 10m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d5285a40b547c4b26c63b88c780cf0 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/3d75f770997143478316aae699474745 : SUCCESS in 59m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5fc0953f90f4797b4cde8f8e8538e44 : SUCCESS in 1h 23m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3dcb5008aaa44394871e5b25f54f681a : FAILURE in 12m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e90ad415969486aa5692aacbcfb7520 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3edd979eddc446768ca16cb66557eaa4 : SUCCESS in 5m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c06ab108c294528bb847ffa9bee3016 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1631d9bb5be84470bd331cc2476f305e : SUCCESS in 46m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2414f76ad80545c39d499a6d7455bad5 : FAILURE in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f7c853518de41a18af3fddae323ac71 : SUCCESS in 1h 06m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5d10abc3c2c74d4988fe30d75d122054 : FAILURE in 1h 18m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4ff800c4573b4e4cb0cafcc84c132c0d : SUCCESS in 1h 41m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6341214efa8044f58f5ebd50a0399041 : SUCCESS in 1h 06m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/78b61cf771b8464c8d57465aa3317d9c : FAILURE in 1h 18m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cd073da2d593423190adf6c1722bfeb6 : SUCCESS in 1h 24m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46c956ac79bc4e9e91379876423661e3 : SUCCESS in 2h 08m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32c0af43e0d846feab6734d9d81889ad : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e9fb58fb104a4a5a98e3074bf110565739330b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"date":"2020-12-17 09:07:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"311b6d5bd40491e4af49540dd5cbc7b7a3e61b92","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 09:38:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/40/767440/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/21742ea/21742ea.html : FAILURE in 17m 50s","accounts_in_message":[],"_revision_number":4},{"id":"1f168a4ec29c06adef5fb916df39ff53ac60daea","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-17 09:50:01.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767440/767440/4/check-not-vote/ext-cinder-zuul/110b218 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":4},{"id":"339fdac09f59f40bc68d1dace07ec11b5d08fdb2","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 10:02:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-primera 3par-iscsi-driver-primera : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"5116840890c6d42987d3071c27b80d86fbf3f7ad","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-17 10:29:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/40/767440/4/check/dsvm-tempest-cisco-zonemanager-job/7786afa/ : SUCCESS in 1h 17m 45s","accounts_in_message":[],"_revision_number":4},{"id":"a3fa72df425728f8d3c09277626fd968eb347fae","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-17 10:38:34.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767440/4/iscsi/ : FAILURE in 1h 30m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/767440/4/smb/ : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":4},{"id":"54f615a62611235110ec7452f54ad9cb0437d8bb","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-17 10:48:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/40/767440/4/ : SUCCESS in 1h 39m 56s","accounts_in_message":[],"_revision_number":4},{"id":"0b096619571bc72e276ddd23b61a444bdf2dffbd","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-17 11:47:02.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/3/20201217163316/NEC-ISCSI : SUCCESS in 2h 06m 51s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/3/20201217163316/NEC-FC : FAILURE in 4h 13m 43s","accounts_in_message":[],"_revision_number":3},{"id":"58b663ff48bf354670143c2675e852bf21292f98","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-17 12:18:11.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-4/ns5_nfs : FAILURE in 53m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-4/ns5_iscsi : SUCCESS in 3h 10m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"8263bc23548a64f6abb30f3a41bb4c2be57a9084","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-17 12:24:47.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/40/767440/4/check/DellEMC_Unity_FC/9f2a2b4/DellEMC_Unity_FC/320 : SUCCESS in 1h 19m 00s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/40/767440/4/check/DellEMC_Unity_iSCSI/2646fe8/DellEMC_Unity_iSCSI/303 : SUCCESS in 1h 48m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c7fac4896a09d07f6ba9f10a65593cb61bd4cbb8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-17 12:56:43.000000000","message":"Patch Set 4:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/40/767440/4/check/dsvm-tempest-zadara-driver/d99e304 : FAILURE in 1h 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7343522e780510b68cc29dc557bda847c45dad8d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-17 13:26:31.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/40/767440/4/check/tempest-dsvm-full-focal-py3/05c7ea4/ : SUCCESS in 2h 06m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/40/767440/4/check/devstack-plugin-ceph-dsvm-focal-py3/a7bd43f/ : SUCCESS in 1h 50m 33s","accounts_in_message":[],"_revision_number":4},{"id":"04e2cc3e3f3a2a1dc2f5e663ad7f09dbe30403a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 13:57:18.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d628eab53b2e4617ae70dd29de4d003f : FAILURE in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a8f85c34d3f491cb4d72144aa811e7d : FAILURE in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9676fb9da9ee4f16a802f467c9cde517 : FAILURE in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6af5fa48ae3f47abbe5e5597407a84ab : FAILURE in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d58875a287844633b05bc378dfa8061c : FAILURE in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4791cad045e64bb694487d487f4db027 : SUCCESS in 18m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/d8bdec94761b4e1cb8a091dfab2bd664 : SUCCESS in 1h 06m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28b0e9c9e91340ab98df7d0e32df5fe4 : SUCCESS in 1h 16m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eb661baa874941c097afd016a3a2d807 : FAILURE in 10m 32s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ec2f53204a204e7fa8288be09e11d771 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5d1b038235e4654a995664f70df3d69 : SUCCESS in 6m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1905bc316efa45478aee1e78c050e4f1 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/11a128e351384a7e9f35bf8c0aa20389 : SUCCESS in 37m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3aa402c0cdb94d09852286a38ede0cf4 : FAILURE in 4m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2c5cf6125e834842be8950d8ee6e16dd : SUCCESS in 1h 06m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ed3003a70f914088977209604d5f58b4 : FAILURE in 1h 10m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/74f397be18dd4a0db5f946d1d52c854c : SUCCESS in 1h 31m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2eaa0681f07b451ea158f629127fc6a1 : SUCCESS in 1h 08m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fa359e40b10d4842b60a55bcb0775602 : SUCCESS in 1h 05m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1c5bfc9958ce470b85e33acdb22413d1 : SUCCESS in 1h 43m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d583665bd4d47bc8c3dd3dc2adfe803 : SUCCESS in 1h 58m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2aa2adab23de4dfcb22f176d7a5b300e : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":4},{"id":"557dbce08c23fbaf3660e469313e2f34e7a6bf16","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-17 21:48:09.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/40/767440/4/check/DellEMC_XtremIO_FC/81617d9/DellEMC_XtremIO_FC/338 : SUCCESS in 1h 42m 54s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/40/767440/4/check/DellEMC_XtremIO_iSCSI/b77a55d/DellEMC_XtremIO_iSCSI/333 : SUCCESS in 3h 26m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"8229f3644b0fe6c83919e3c36919e1b51fef7fd2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-17 22:41:31.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/40/767440/4/check/DellEMC_PowerMAX_FC/a699477/DellEMC_PowerMAX_FC/272 : FAILURE in 4h 26m 20s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/40/767440/4/check/DellEMC_PowerMAX_iSCSI/8d752f2/DellEMC_PowerMAX_iSCSI/248 : FAILURE in 5h 30m 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"99065968529f77a5115527cb91f7c96239600c7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"date":"2020-12-18 01:29:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1dccfab170af330760660f753adf72f5c87ad154","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-18 02:13:37.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767440/767440/5/check-not-vote/ext-cinder-zuul/07ebbc4 : SUCCESS in 43m 16s","accounts_in_message":[],"_revision_number":5},{"id":"254c119b3bc4437825e43fb9e08bfc9bff95922e","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-18 02:35:40.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767440/5/iscsi/ : FAILURE in 1h 05m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/767440/5/smb/ : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":5},{"id":"97d906b0d7e46d9429dd424f60ce4b690b7992c1","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-18 02:52:12.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/4/20201218074841/NEC-ISCSI : SUCCESS in 2h 15m 02s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/4/20201218074841/NEC-FC : FAILURE in 4h 03m 29s","accounts_in_message":[],"_revision_number":4},{"id":"b41660a9405d8b7b33191f826dd99d240a40ae89","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-18 02:53:14.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/40/767440/5/check/dsvm-tempest-cisco-zonemanager-job/0aa5cc5/ : SUCCESS in 1h 19m 03s","accounts_in_message":[],"_revision_number":5},{"id":"44fc4991eb767847748b38a529fdc7b87eb496be","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-18 03:29:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/40/767440/5/ : SUCCESS in 1h 27m 05s","accounts_in_message":[],"_revision_number":5},{"id":"6a36a5e1a360e2c92769c60f643d533fcbec1188","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-18 03:36:13.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/40/767440/5/check/tempest-dsvm-full-focal-py3/4c0a6c7/ : SUCCESS in 1h 37m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/40/767440/5/check/devstack-plugin-ceph-dsvm-focal-py3/7c20d8b/ : SUCCESS in 1h 57m 40s","accounts_in_message":[],"_revision_number":5},{"id":"134dde7cb4b0e37421e5afc18b1869beb9e8137b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 04:04:49.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd54928034f64b53bd29203306d6a09f : FAILURE in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c982db99d14d3ca0dcc4ff60f9eae9 : FAILURE in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70d964a8d11f4af3a340df1cc84f4413 : FAILURE in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02dad057e13745f7bd9f81694465ebec : FAILURE in 9m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e01f07f4b3c4a91895f1df508c48bf6 : FAILURE in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cfa279ed260441e8bf8fb903eff779b : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/144e40c7e6f444fea14ed77fd465fef2 : SUCCESS in 59m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f77a1fabb9a8421f9d9d72198c72d5f6 : FAILURE in 1h 01m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d326ff1211e47cd8fdf7c997a573e49 : FAILURE in 9m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69f517c623b44936af1116ec21c81e89 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/272aec1e213947e099b22b7ea35695d7 : SUCCESS in 7m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea85146931f945279fb797fd70b83834 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c8a533c587d848f1ad8e9ebe61c5ece7 : SUCCESS in 38m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bc21e32ec9314d0a97f412a522390ad9 : FAILURE in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f3a369b1330474aafa067f829aabccf : SUCCESS in 1h 06m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a1a7f348077e4d6aad828b27ced85e64 : FAILURE in 1h 30m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/92ab190749ba44c7adfc73fbdad2b107 : SUCCESS in 1h 44m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a3e7992b060c47ff9d51a57327413f32 : SUCCESS in 1h 01m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/47ea384aa4164e2ca875f0ab111998c6 : SUCCESS in 1h 20m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ec9228ca75d47c6807ff14bca894c49 : SUCCESS in 1h 24m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cf430d2d8a044814bf71d6c4c7b3455c : SUCCESS in 2h 27m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61cda534ae174280868b5faec66568f0 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":5},{"id":"890cf486917c7bcf9fd1a5fe3afa899b48704b30","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-18 04:36:18.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-5/ns5_nfs : SUCCESS in 3h 6m 14s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-5/ns5_iscsi : SUCCESS in 3h 3m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"8886aec7d837d527df3d56987853400676b1c500","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-18 05:04:39.000000000","message":"Patch Set 5:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/40/767440/5/check/dsvm-tempest-zadara-driver/7b429ca : SUCCESS in 1h 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b9e6b6c5f408345a508f51025988a2ccbdf9c5a3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-18 06:42:37.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/40/767440/5/check/DellEMC_Unity_FC/00f91e0/DellEMC_Unity_FC/333 : SUCCESS in 1h 10m 48s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/40/767440/5/check/DellEMC_Unity_iSCSI/f1ecfc7/DellEMC_Unity_iSCSI/312 : SUCCESS in 1h 45m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"c094b0edd699f3e2dcbf41ebf472e2f56c21eb0a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-18 23:02:47.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/5/20201219031727/NEC-ISCSI : SUCCESS in 4h 25m 56s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/5/20201219031727/NEC-FC : FAILURE in 4h 45m 19s","accounts_in_message":[],"_revision_number":5},{"id":"2c2230211b79e20de0987dcaea32ee545b81631e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-18 23:43:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/40/767440/5/check/DellEMC_XtremIO_FC/24cc1e8/DellEMC_XtremIO_FC/346 : SUCCESS in 1h 40m 48s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/40/767440/5/check/DellEMC_XtremIO_iSCSI/f6b97ac/DellEMC_XtremIO_iSCSI/339 : SUCCESS in 4h 05m 54s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"54bde98f2d4d5d244b587c5c79c9bf1d17b2bbf0","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-19 03:06:34.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/40/767440/5/check/DellEMC_PowerMAX_FC/4361eeb/DellEMC_PowerMAX_FC/278 : SUCCESS in 4h 25m 01s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/40/767440/5/check/DellEMC_PowerMAX_iSCSI/b8d0a80/DellEMC_PowerMAX_iSCSI/254 : SUCCESS in 4h 45m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"cf34161df431ba5a2489bb87ca6f0c954c2875a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"date":"2020-12-19 03:39:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0e7797eadc0590f9223faa5749fc27776fe48f78","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-19 04:29:55.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/767440/767440/6/check-not-vote/ext-cinder-zuul/103cb8b : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":6},{"id":"5cb78c0ddf5553d31240c986609c60da9cbeb6d9","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-19 04:44:47.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767440/6/iscsi/ : FAILURE in 1h 04m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/767440/6/smb/ : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":6},{"id":"f595af2d0bfd4a794e7d56675fa559b83cea99c6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-19 04:55:21.000000000","message":"Patch Set 6:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/40/767440/6/check/dsvm-tempest-zadara-driver/0655d05 : SUCCESS in 1h 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"03626e6f408caba6d9c9614e9e0752c9a885244f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-19 05:03:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/40/767440/6/check/dsvm-tempest-cisco-zonemanager-job/6519ab0/ : SUCCESS in 1h 19m 02s","accounts_in_message":[],"_revision_number":6},{"id":"ca3bc8b24359cf7afa22ef9a31ca2ddd196442c1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-19 05:22:26.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/40/767440/6/check/tempest-dsvm-full-focal-py3/19798d7/ : SUCCESS in 1h 37m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/40/767440/6/check/devstack-plugin-ceph-dsvm-focal-py3/798d979/ : FAILURE in 26m 28s","accounts_in_message":[],"_revision_number":6},{"id":"81b4f51a2236e5876273e908475d429466e6b9cc","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-19 05:42:18.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/40/767440/6/ : SUCCESS in 1h 44m 59s","accounts_in_message":[],"_revision_number":6},{"id":"af758e0927fdf19b8adc46d2150bf72cc600afaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 05:56:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31d598fdef5c4ff9b0b164a559027603 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee7c79f31f44e16b0c41f37084d8cfa : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfee3feca9674e378b961be83ee660b8 : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95d940662364430296594966038bdf15 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d150d49ddcf34f78a50282490140c5a0 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa698e943b27486bb2e959508b17dde2 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/a473ff5bd5e04689b5b6cd85882f50d0 : SUCCESS in 52m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0684d37895114e4da7d3295eadedd2b6 : SUCCESS in 1h 08m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e899a9970040411b8807496ab6294d45 : SUCCESS in 12m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01e76d88df3a4655b89b43a2d2030891 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a47632a57710403b9b6da4fd90a82516 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a35ba36e5454ba298cbf67b71db8d1c : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/48fcf6cc3e3c414db68d4affbd91f9c9 : SUCCESS in 42m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3afd0b8639ba4fc19cc3f18515d77ab8 : FAILURE in 4m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d190b7fa4e3b4d01a35184572982d605 : SUCCESS in 1h 25m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8848068c35c648cab3e04c8e5ef11692 : FAILURE in 1h 20m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/95ec8ed04fc84ab48d4e06555bffc9dd : SUCCESS in 1h 44m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c55379bb9096405b83c60dbcf173d3cf : SUCCESS in 56m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9d73878bec764c838a0b2b4d173e7e64 : SUCCESS in 1h 30m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/21026657a00c4927929e400c5832106a : SUCCESS in 1h 27m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b430033fa85433189fc19782f3bb4fe : SUCCESS in 2h 15m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f14fd521ff7646c9ae7a67d3c8f17364 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":6},{"id":"895fb7743be7479cb483765e9a8b52f5d68d7fb4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-19 06:46:39.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-6/ns5_nfs : SUCCESS in 3h 6m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-6/ns5_iscsi : SUCCESS in 3h 5m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"811be0b3d1624c0e614b6bd1d3b2aceb0e9bfbc2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-12-19 07:12:50.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/40/767440/6/check/DellEMC_PowerStore_iSCSI/a34938a/DellEMC_PowerStore_iSCSI/191 : SUCCESS in 1h 15m 26s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/40/767440/6/check/DellEMC_PowerStore_FC/cc98c4d/DellEMC_PowerStore_FC/187 : SUCCESS in 1h 09m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"df4b7a3f0bc9b8799b320ada4bcec8056ad715a6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-19 07:24:35.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/40/767440/6/check/DellEMC_Unity_FC/ef708a7/DellEMC_Unity_FC/341 : SUCCESS in 1h 23m 51s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/40/767440/6/check/DellEMC_Unity_iSCSI/f051512/DellEMC_Unity_iSCSI/319 : SUCCESS in 1h 39m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"8150aca777f827ef09e4a534e36648016b0177c3","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2020-12-19 07:59:54.000000000","message":"Patch Set 6:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/767440-6/Seagate-iSCSI : SUCCESS in 46:02.87\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/767440-6/Seagate-FC : SUCCESS in 40:54.04","accounts_in_message":[],"_revision_number":6},{"id":"9605b320c59e5550cb75fb371c1fdd915d8ed563","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2020-12-19 08:08:03.000000000","message":"Patch Set 6:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/767440-6/PVME-iSCSI : SUCCESS in 49:28.42\n* PVME-FC http://os-logs.tristero.net/dell-ci/767440-6/PVME-FC : SUCCESS in 48:15.26","accounts_in_message":[],"_revision_number":6},{"id":"4fe0f9c0bda7f8fc66069aa372d4300fa4512d9b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-12-19 08:12:49.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/40/767440/6/check/DellEMC_SC_FC/9b3fb48/DellEMC_SC_FC/139 : SUCCESS in 2h 15m 30s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/40/767440/6/check/DellEMC_SC_iSCSI/c9e9d4a/DellEMC_SC_iSCSI/137 : SUCCESS in 1h 36m 30s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"7c2ec21f5907cec19e2acf917f0b9f43876d65d8","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2020-12-19 08:13:12.000000000","message":"Patch Set 6:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/767440-6/Lenovo-iSCSI : SUCCESS in 54:35.36\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/767440-6/Lenovo-FC : SUCCESS in 46:10.76","accounts_in_message":[],"_revision_number":6},{"id":"b11a5386124187350cd681fad42dcee518007910","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-12-19 08:16:56.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/40/767440/6/check/DellEMC_PowerFlex/5b15ea1/DellEMC_PowerFlex/151 : SUCCESS in 2h 19m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"b4ca745da53e0a81c4fc37fa556ecf101bd5bbf1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-19 08:50:32.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/40/767440/6/upstream-check/cinder-cDOT-iSCSI/205f020/ : SUCCESS in 38m 36s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/40/767440/6/upstream-check/cinder-cDOT-iSCSI-IPv6/8114c8f/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/40/767440/6/upstream-check/cinder-cDOT-NFS/a0b27f2/ : SUCCESS in 37m 59s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/40/767440/6/upstream-check/cinder-cDOT-NFS-IPv6/8d5d7a4/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/40/767440/6/upstream-check/cinder-cDOT-FCP/5e9deca/ : SUCCESS in 1h 02m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"790618913170e433ed16e4a3985183818b04cf09","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-19 08:56:35.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/40/767440/6/check/DellEMC_VNX_iSCSI/18bac91/DellEMC_VNX_iSCSI/194 : SUCCESS in 2h 59m 18s\n- DellEMC_VNX_FC http://publiclogs.emc.com/40/767440/6/check/DellEMC_VNX_FC/b5c95df/DellEMC_VNX_FC/195 : SUCCESS in 1h 06m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"5e548fa8b8490eddc382875724da432ffec707fc","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2020-12-19 09:28:27.000000000","message":"Patch Set 6:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/767440-6/HPMSA-iSCSI : FAILURE in 1:28:49\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/767440-6/HPMSA-FC : FAILURE in 1:27:07","accounts_in_message":[],"_revision_number":6},{"id":"fb575e34896dd1cdc33861dad3b35c6b478e3e55","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-12-19 10:03:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/40/767440/6/upstream-solidfire-check/cinder-solidfire-iSCSI/21dd011/ : FAILURE in 1h 50m 42s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"9d3dc250c4e8ca888fae2b3a5e706770cb45b9be","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-19 14:17:02.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/6/20201219183917/NEC-ISCSI : SUCCESS in 1h 58m 34s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/767440/6/20201219183917/NEC-FC : FAILURE in 4h 37m 43s","accounts_in_message":[],"_revision_number":6},{"id":"fb8870c0667aaf41c114623c1a88955c160deea1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-19 17:10:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/40/767440/6/check/DellEMC_XtremIO_FC/944ce9c/DellEMC_XtremIO_FC/353 : FAILURE in 4m 14s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/40/767440/6/check/DellEMC_XtremIO_iSCSI/42c3f70/DellEMC_XtremIO_iSCSI/342 : FAILURE in 4h 55m 31s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"6b559bfc357c265621e9144dc1519d9a1e0a162e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-19 17:30:00.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/40/767440/6/check/DellEMC_PowerMAX_FC/99029ac/DellEMC_PowerMAX_FC/282 : SUCCESS in 4h 27m 59s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/40/767440/6/check/DellEMC_PowerMAX_iSCSI/e63fd95/DellEMC_PowerMAX_iSCSI/257 : FAILURE in 4h 21m 15s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"48a0a2ed2f18956e4897224948e4ca261ea0fe0a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2020-12-21 11:50:27.000000000","message":"Patch Set 6:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/767440/6/fc : SUCCESS in 0h 47m 23s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/767440/6/iscsi : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":6},{"id":"29ca9fe8b80e5e8da117f3b338443df424384370","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-22 03:59:53.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-6/ns5_nfs : SUCCESS in 3h 1m 9s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-40-767440-6/ns5_iscsi : SUCCESS in 3h 2m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"f7fa52e7e82b3075fd87175f91e35dbf2a779076","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-25 11:13:35.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f923494c81b7e1a16abe7bcb77af8e2d8e98000e","revisions":{"81e65eef8cecc2dc2bc6207c9ad0fd1e86e6d99c":{"kind":"REWORK","_number":1,"created":"2020-12-17 01:07:37.000000000","uploader":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"ref":"refs/changes/40/767440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/767440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/767440/1"}}},"commit":{"parents":[{"commit":"f8e70fd31dd165346d87ea43ae324226d1f65551","subject":"Merge \"Make test-setup.sh compatible with mysql8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e70fd31dd165346d87ea43ae324226d1f65551"}]}],"author":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-16 10:29:28.000000000","tz":480},"committer":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-16 10:29:28.000000000","tz":480},"subject":"backup dependents failure","message":"backup dependents failure\n\nIncremental backup created fails, its parent_id will record the id of the\nparent backup, but num_dependent_backup is not increased by 1, resulting in\nhas_dependent_backups being false, but the deletion of the parent backup\nwill fails. This patch adds 1 to the num_dependent_backup of the parent\nbackup when the backup creation fails.\n\nChange-Id: I97afdc561b346abb87c3801aaab4b9f3e7492d6d\nCloses-Bug: #1907542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81e65eef8cecc2dc2bc6207c9ad0fd1e86e6d99c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81e65eef8cecc2dc2bc6207c9ad0fd1e86e6d99c"}]},"branch":"refs/heads/master"},"0bd31737f5391f272c4f452d3f33bf108d00edcb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-17 02:11:03.000000000","uploader":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"ref":"refs/changes/40/767440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/767440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/767440/2"}}},"commit":{"parents":[{"commit":"f8e70fd31dd165346d87ea43ae324226d1f65551","subject":"Merge \"Make test-setup.sh compatible with mysql8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e70fd31dd165346d87ea43ae324226d1f65551"}]}],"author":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-16 10:29:28.000000000","tz":480},"committer":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-17 02:06:41.000000000","tz":480},"subject":"backup dependents failure","message":"backup dependents failure\n\nIncremental backup created fails, its parent_id will record the id of the parent backup, but num_dependent_backup is not increased by 1, resulting in has_dependent_backups being false, but the deletion of the parent backup\nwill fails. This patch adds 1 to the num_dependent_backup of the parent\nbackup when the backup creation fails.\n\nChange-Id: I97afdc561b346abb87c3801aaab4b9f3e7492d6d\nCloses-Bug: #1907542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0bd31737f5391f272c4f452d3f33bf108d00edcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0bd31737f5391f272c4f452d3f33bf108d00edcb"}]},"branch":"refs/heads/master"},"5a6dbeec391af1931589aad872de8f75fd74318c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-17 02:44:18.000000000","uploader":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"ref":"refs/changes/40/767440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/767440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/767440/3"}}},"commit":{"parents":[{"commit":"f8e70fd31dd165346d87ea43ae324226d1f65551","subject":"Merge \"Make test-setup.sh compatible with mysql8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e70fd31dd165346d87ea43ae324226d1f65551"}]}],"author":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-16 10:29:28.000000000","tz":480},"committer":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-17 02:11:37.000000000","tz":480},"subject":"backup dependents failure","message":"backup dependents failure\n\nIncremental backup created fails, its parent_id will record the id of the parent backup, but num_dependent_backup is not increased by 1, resulting in has_dependent_backups being false, but the deletion of the parent backup will fails. This patch adds 1 to the num_dependent_backup of the parent backup when the backup creation fails.\n\nChange-Id: I97afdc561b346abb87c3801aaab4b9f3e7492d6d\nCloses-Bug: #1907542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a6dbeec391af1931589aad872de8f75fd74318c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a6dbeec391af1931589aad872de8f75fd74318c"}]},"branch":"refs/heads/master"},"bc213fe14f45bfd1cb80cb553ee95e36b5b7bd0c":{"kind":"REWORK","_number":4,"created":"2020-12-17 09:07:02.000000000","uploader":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"ref":"refs/changes/40/767440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/767440/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/767440/4"}}},"commit":{"parents":[{"commit":"f8e70fd31dd165346d87ea43ae324226d1f65551","subject":"Merge \"Make test-setup.sh compatible with mysql8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e70fd31dd165346d87ea43ae324226d1f65551"}]}],"author":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-16 10:29:28.000000000","tz":480},"committer":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-17 09:02:50.000000000","tz":480},"subject":"backup dependents failure","message":"backup dependents failure\n\nIncremental backup created fails, its parent_id will record the id of the parent backup, but num_dependent_backup is not increased by 1, resulting in has_dependent_backups being false, but the deletion of the parent backup will fails. This patch adds 1 to the num_dependent_backup of the parent backup when the backup creation fails.\n\nChange-Id: I97afdc561b346abb87c3801aaab4b9f3e7492d6d\nCloses-Bug: #1907542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc213fe14f45bfd1cb80cb553ee95e36b5b7bd0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc213fe14f45bfd1cb80cb553ee95e36b5b7bd0c"}]},"branch":"refs/heads/master"},"9cc7116607d8df89af5e351ea23135b42d216ae1":{"kind":"REWORK","_number":5,"created":"2020-12-18 01:29:24.000000000","uploader":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"ref":"refs/changes/40/767440/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/767440/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/767440/5"}}},"commit":{"parents":[{"commit":"f8e70fd31dd165346d87ea43ae324226d1f65551","subject":"Merge \"Make test-setup.sh compatible with mysql8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e70fd31dd165346d87ea43ae324226d1f65551"}]}],"author":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-16 10:29:28.000000000","tz":480},"committer":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-18 01:27:22.000000000","tz":480},"subject":"backup dependents failure","message":"backup dependents failure\n\nIncremental backup created fails, its parent_id will record the id of the parent backup, but num_dependent_backup is not increased by 1, resulting in has_dependent_backups being false, but the deletion of the parent backup will fails. This patch adds 1 to the num_dependent_backup of the parent backup when the backup creation fails.\n\nChange-Id: I97afdc561b346abb87c3801aaab4b9f3e7492d6d\nCloses-Bug: #1907542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cc7116607d8df89af5e351ea23135b42d216ae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cc7116607d8df89af5e351ea23135b42d216ae1"}]},"branch":"refs/heads/master"},"f923494c81b7e1a16abe7bcb77af8e2d8e98000e":{"kind":"REWORK","_number":6,"created":"2020-12-19 03:39:06.000000000","uploader":{"_account_id":30754,"name":"FengJiankui","email":"fengjiankui@inspur.com","username":"FengJiankui"},"ref":"refs/changes/40/767440/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/767440/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/767440/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/767440/6"}}},"commit":{"parents":[{"commit":"f8e70fd31dd165346d87ea43ae324226d1f65551","subject":"Merge \"Make test-setup.sh compatible with mysql8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8e70fd31dd165346d87ea43ae324226d1f65551"}]}],"author":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-16 10:29:28.000000000","tz":480},"committer":{"name":"FengJiankui","email":"fengjiankui@inspur.com","date":"2020-12-19 03:31:54.000000000","tz":480},"subject":"backup dependents failure","message":"backup dependents failure\n\nIncremental backup created fails, its parent_id will record the id of the parent backup, but num_dependent_backup is not increased by 1, resulting in has_dependent_backups being false, but the deletion of the parent backup will fails. This patch adds 1 to the num_dependent_backup of the parent backup when the backup creation fails.\n\nChange-Id: I97afdc561b346abb87c3801aaab4b9f3e7492d6d\nCloses-Bug: #1907542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f923494c81b7e1a16abe7bcb77af8e2d8e98000e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f923494c81b7e1a16abe7bcb77af8e2d8e98000e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
