)]}'
{"id":"openstack%2Fcinder~767477","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~I00f1bd46bc8715591016f59c32651fc2c9cddf35","project":"openstack/cinder","branch":"stable/victoria","topic":"bug/1890591-stable/victoria","hashtags":[],"change_id":"I00f1bd46bc8715591016f59c32651fc2c9cddf35","subject":"[SVF]:Reduce slowness by caching pool information","status":"MERGED","created":"2020-12-17 08:25:48.000000000","updated":"2021-02-05 20:01:38.000000000","submitted":"2021-02-05 19:59:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767477-bug/1890591-stable/victoria","cherry_pick_of_change":746267,"cherry_pick_of_patch_set":5,"meta_rev_id":"81068fbea82c46fbcd3f1d722d6ae0eddaaf8031","_number":767477,"virtual_id_number":767477,"owner":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 19:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"all":[{"value":1,"date":"2021-01-27 02:16:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-01-28 10:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-25 18:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2021-01-27 08:49:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-05 18:21:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 08:34:33.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"state":"REVIEWER"},{"updated":"2020-12-17 08:55:26.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 09:15:39.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 09:37:14.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-17 09:50:23.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-17 10:15:45.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 10:19:57.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 11:21:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 11:33:05.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 22:48:15.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-20 15:29:32.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2020-12-20 16:31:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-22 12:08:49.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"CC"},{"updated":"2021-01-25 18:15:43.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-01-27 02:16:24.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-01-27 08:49:25.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-01-28 10:12:42.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"}],"messages":[{"id":"f9bf53ec03adca6f6eb590cdad9fd186fcca62ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-17 08:25:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"30b860013e7c8652b95ba8d4a4133c23a4abad6f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-17 08:55:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/77/767477/1/ : RETRY_LIMIT in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"654269249778fea20e4d6a2b0534d175cc3866b3","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 09:15:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/77/767477/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/2cff743/2cff743.html : FAILURE in 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"12e525fa97f91f9e350f0d417e9efb1d4d95620a","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 09:16:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/77/767477/1/check/3par-iscsi-driver-primera/974858f/974858f.html : FAILURE in 50m 04s","accounts_in_message":[],"_revision_number":1},{"id":"209f42191293f2ab2c66d882f6f9494a165004b8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-17 09:37:14.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767477/1/iscsi/ : FAILURE in 1h 10m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/767477/1/smb/ : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":1},{"id":"77b949a88599abbe30f3f7f993ea20f899885eab","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-17 09:50:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/77/767477/1/check/dsvm-tempest-cisco-zonemanager-job/2ed3542/ : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0d2da91097350fd40bc61a5e5cdb4e14d4f457d9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-17 10:15:45.000000000","message":"Patch Set 1:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/77/767477/1/check/dsvm-tempest-zadara-driver/111c7c9 : FAILURE in 1h 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e43cd247c55b453051cbf3f513d4a986be6091d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-17 10:19:57.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/77/767477/1/check/tempest-dsvm-full-focal-py3/17a74c6/ : SUCCESS in 1h 38m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/77/767477/1/check/devstack-plugin-ceph-dsvm-focal-py3/d64d22f/ : FAILURE in 55m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a280eb489f09389b244966b356aca0ba1ca26a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 11:21:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ed5e3ba39fa45d4b7ff99aa294e2274 : FAILURE in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab8b8838917b4402b3b21f74f8ad1aff : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3efd9b23f81846cfa0537b13252553a1 : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12cca6616dcd41fa96eed9ea3e66e490 : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28221923b204e3e849602f6d2ef1c1b : SUCCESS in 14m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/b68c597c7a294c6d816feec958707443 : SUCCESS in 1h 20m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/771dc25fe03348229b34a8c1831890df : SUCCESS in 1h 09m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47ac4d5641da4403b6da093662433b62 : SUCCESS in 6m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0dbc91a3f7d344f5ad2ff41ec7f9a630 : SUCCESS in 14m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e35eca729056477e85cd608eae7c9627 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6097ac2c406f4ccf97ec9d65afa7491e : SUCCESS in 6m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc8388af76014fdc80b66672faa2e00c : SUCCESS in 7m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4566a10ab3894ca29b66dcd7c80ce0a6 : SUCCESS in 43m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7715e3332b9e451ab26c1e73843fbb52 : FAILURE in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e1161565632244a988af50ec72174c14 : SUCCESS in 1h 10m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1e97b58bf5d2412289478c5c356f512d : FAILURE in 1h 34m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cc7bbd4f64f94cf19abd7042dadb4389 : FAILURE in 1h 59m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/31a5348e3afd4ebb9d5c2734cd86cdfa : SUCCESS in 1h 10m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e8d03986f659414e94675e3ff44094dd : SUCCESS in 1h 26m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/725febaad9ce4d14bca15f18a59e8785 : SUCCESS in 1h 25m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26255f30c41d40b3a3e206c215736f67 : SUCCESS in 2h 11m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/617f2574451d46b0ba1830845c468923 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3fe1093721bc691899b1f4c63ba7ca676dde42d8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-17 11:33:05.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-77-767477-1/ns5_nfs : SUCCESS in 3h 6m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-77-767477-1/ns5_iscsi : SUCCESS in 3h 3m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"bdbf086ac20ea48ac545fd4150f65d6fa0ec244b","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-17 22:48:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/77/767477/1/20201218053639/NEC-ISCSI : SUCCESS in 2h 02m 48s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/77/767477/1/20201218053639/NEC-FC : SUCCESS in 2h 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a5e1e0b86c962c80c121f3f97a52a0b9953141e9","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-20 09:09:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3c5729b09fe1d3798cb388fd8cbbd3946119a62","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-20 10:15:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767477/1/iscsi/ : SUCCESS in 1h 04m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/767477/1/smb/ : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b7ab381bf68a1a9a9168dc25a3dd9f170bfb80e8","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-20 10:44:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"69c3485722c51281e33c80c66decc90a4fc089e2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-20 10:50:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/77/767477/1/check/dsvm-tempest-cisco-zonemanager-job/b73c16a/ : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"33f8e960489f45b98def02fc9e25684b0999e2c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 11:10:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d1d45f4d10f46418b319dcb9445fcfd : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bda451b4a574ddf805d3ef1ac10413c : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9bef037bd9749fabd436b5ca2a9aab3 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca922c55e9274908b65f91503a1c66f6 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51f9d5bf3d2f4d67bfb4c329f0ebc99e : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/ea47bf4baf8046d6ace836236cbd3451 : SUCCESS in 56m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f74e6a7eb9054540871fe13c8c8f58d3 : FAILURE in 1h 24m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7a51f7ad563412bb77e3ae2e2656f1a : SUCCESS in 6m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0b1d34f864d84a3ea572e9e898a0e16f : SUCCESS in 10m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc843ac54f3d4fa2b23eb4a91f33ba26 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c0c40f340554147ba7654262e47413f : SUCCESS in 5m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a68d35ea2e94b9dba3972d644bc389e : SUCCESS in 4m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/461c0da7efff45dabf293b34be234c60 : SUCCESS in 50m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68f10d65e951461ab923ae18539e5e5b : FAILURE in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a0f7cbf9fd8946ae9019846f5191050f : SUCCESS in 1h 22m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4bb131ba886a402790290f393e10ab36 : FAILURE in 1h 00m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f68cbe21754c47f6a0863a95044f0ec6 : SUCCESS in 1h 16m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ff55fabdbda4aeeabce2796ad3da75c : SUCCESS in 1h 10m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/96560c62a87f4dfb8055eb5f3379f2b9 : SUCCESS in 1h 18m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ff3c26346bcf4c18bc3ea97b1244ae7c : SUCCESS in 1h 19m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7f7831c60c14781810e8259e18fe0fc : SUCCESS in 1h 59m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2b3470b173540a9a06f4cf791e51892 : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f0e2e3edd30aef1c254e9451548b28fc28cbb214","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-20 11:12:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f648bdf4a933c0aaca1aec1e6f29aeeacc1e4754","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-20 11:48:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767477/1/iscsi/ : SUCCESS in 1h 03m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/767477/1/smb/ : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"16e5639cc287cbc971606087e4ced0790d81c8a2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-20 11:56:19.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/77/767477/1/check/tempest-dsvm-full-focal-py3/2e9a0ad/ : SUCCESS in 1h 26m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/77/767477/1/check/devstack-plugin-ceph-dsvm-focal-py3/704463d/ : FAILURE in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6c2e991f702f99c09c0f313519ae8c752a3bb28b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 13:36:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dba83cc3a2464acf89c46dbe88c0d011 : SUCCESS in 13m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87dd91e3a3e5412f9479315e0b595f61 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61553d325f1f4585bcb99d947472151d : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50105c3a4e02438da831380207996591 : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77f799f2e0f4eefb165f9ebe39a8b41 : SUCCESS in 28m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/5d3dcf94ff8b4d3098768ef083ffea0e : SUCCESS in 59m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d12ebe7c93e84012910449468e2ed637 : SUCCESS in 45m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/322e25dc653b418684e42dd618e49e4c : SUCCESS in 5m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1a14690dfc264f6eb1ed3606b9210fe6 : SUCCESS in 12m 02s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e149188907a4117b3123ad7171f699e : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ac19e51f28d464190ab9b44b3c7cf68 : SUCCESS in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4cd290541eea4ec7bcde29902b22e125 : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cc06f6070a144102a4e1a29ff602e782 : SUCCESS in 41m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/577640cbf4c4471ca82c05e70d246fc7 : FAILURE in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9363bcc61084e289c90c0b916f9ae84 : SUCCESS in 1h 04m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/afd1d354d95441ac8982ed3f7cb7c7e2 : FAILURE in 1h 16m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7d7656083db74080b52b3329e094c0a6 : SUCCESS in 1h 45m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/82e03e1551d141c399f3b1ec40e658e9 : SUCCESS in 1h 08m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1f9a4ac7931a4710b34b5af67b8840ec : SUCCESS in 1h 36m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4842cf53fe02447fa8a05c274f432451 : SUCCESS in 1h 20m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5bf1a0545cc343879f51068342c08e22 : SUCCESS in 2h 20m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fc6664575454aa6a60977c0f314c13d : SUCCESS in 49m 45s","accounts_in_message":[],"_revision_number":1},{"id":"41b57dcb160293464ac2c1cd66e2b03083d27733","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-20 13:41:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/77/767477/1/check/dsvm-tempest-cisco-zonemanager-job/9284be3/ : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e0fb916757df535860bbb22cb1cdeb20ed50a754","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-12-20 15:29:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/77/767477/1/upstream-solidfire-check/cinder-solidfire-iSCSI/9b34e66/ : FAILURE in 1h 51m 14s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe990631d575ac4f5b21db5969ce34c82838ffa6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-20 16:31:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/77/767477/1/upstream-check/cinder-cDOT-iSCSI/fd8082e/ : FAILURE in 32m 50s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/77/767477/1/upstream-check/cinder-cDOT-iSCSI-IPv6/50c51ea/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/77/767477/1/upstream-check/cinder-cDOT-NFS/151eb0f/ : SUCCESS in 37m 19s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/77/767477/1/upstream-check/cinder-cDOT-NFS-IPv6/ae74ce0/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/77/767477/1/upstream-check/cinder-cDOT-FCP/7ca4107/ : FAILURE in 2h 53m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e353949c816f89548d1eb1d3bed742acc4773c7","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-22 08:01:13.000000000","message":"Patch Set 1:\n\nrun-IBM Storage CI","accounts_in_message":[],"_revision_number":1},{"id":"1c2462ad2a960c7c7c6a150195d981af9f2e4206","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-12-22 12:08:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1993/index.html : SUCCESS in 1h 15m 52s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/2139/index.html : SUCCESS in 1h 58m 18s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/2128/index.html : SUCCESS in 1h 31m 28s","accounts_in_message":[],"_revision_number":1},{"id":"056bfb50b78f295c244441cca367930d781401a3","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-01-25 18:15:43.000000000","message":"Patch Set 1: Code-Review+2\n\nPerformance changes can or cannot be appropriate.  I don\u0027t have a big concern with this though.  It is a small change.","accounts_in_message":[],"_revision_number":1},{"id":"3119f4a6c7a8f264c622c592c8ddadf942ec4fad","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-27 02:16:24.000000000","message":"Patch Set 1: Code-Review+1\n\nI\u0027m on the fence about this one.  It\u0027s an optimization, but as Jay points out, it\u0027s a low-risk isolated change.  I\u0027d like some more input from the stable cores about whether this should be backported.","accounts_in_message":[],"_revision_number":1},{"id":"627e492f95144fba7e8ab05a83d4c10344894d7e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-27 08:49:25.000000000","message":"Patch Set 1: Code-Review+1\n\nI agree to the concern of Jay and Brian but the stable review guidelines[1] doesn\u0027t mention blocking/holding similar changes.\nThe patch is merged in master and looks good to backport.\n\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#review-guidelines","accounts_in_message":[],"_revision_number":1},{"id":"40ed8834a29f16d4b50426b4ed9fe74a0be3243c","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-01-28 10:12:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20ee1729a786cd798e67a5fb13a7aabb81387a58","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-05 18:21:55.000000000","message":"Patch Set 1: Workflow+1\n\nGiven the other comments I think we can merge this.","accounts_in_message":[],"_revision_number":1},{"id":"e07c58367f974d34082659d791c46ee233c4a31d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 18:22:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcc101f6987768cc476a73b07044fb9a4a71b3a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 19:59:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20a245f273054619bd7ecf0456ba50c3 : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0855adb53f994215bf3adffd668078ed : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a92e7c3cf364edcb9f86c417d958bc9 : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97e9aa2bc75e42bfb0c09ec0522fe0e2 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9644fb26a50441c38de5e3f2565ad708 : SUCCESS in 15m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/f8c8efe3f05b48b4a63c261eed404b50 : SUCCESS in 57m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5da750db0a3406f8403aa30d88520ba : SUCCESS in 1h 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdcfbad85e864c00ac52bf2ec17a532c : SUCCESS in 5m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd1458903b3d45728bc2ae103069be12 : SUCCESS in 1h 07m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cf50933cb7564ec2bbbcfaa2d81d8924 : SUCCESS in 1h 15m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a40f09cab2d4fa8855117d67e780968 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"476c52466d849c6139dce95d1d1e832206886c8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 19:59:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"81068fbea82c46fbcd3f1d722d6ae0eddaaf8031","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 20:01:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb350abd70894ab9a37b3a727a793be6 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a936f5b0edc4e9fb506a288dc9e0067 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2cddc108b174895b3575e9a2640d3cd8afc21ab","revisions":{"e2cddc108b174895b3575e9a2640d3cd8afc21ab":{"kind":"REWORK","_number":1,"created":"2020-12-17 08:25:48.000000000","uploader":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"ref":"refs/changes/77/767477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/767477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/767477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/767477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/767477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/767477/1"}}},"commit":{"parents":[{"commit":"cdc586631586dc9782794a90f129ce3f34118193","subject":"Fix volume rekey during clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdc586631586dc9782794a90f129ce3f34118193"}]}],"author":{"name":"GirishChilukuri","email":"girish.chilukuri@ibm.com","date":"2020-09-25 20:31:25.000000000","tz":0},"committer":{"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","date":"2020-12-17 08:25:48.000000000","tz":0},"subject":"[SVF]:Reduce slowness by caching pool information","message":"[SVF]:Reduce slowness by caching pool information\n\n[Spectrum Virtualize Family] Data_reduction is a pool specific property,\nthis property should be saved for a specific pool during initialisation.\n\nWith this, calls to get_pool_attrs can be reduced everytime create_vdisk\nis called, in order to check if pool is a data_reduction_pool.\n\ncloses bug: #1890591\n\nChange-Id: I00f1bd46bc8715591016f59c32651fc2c9cddf35\n(cherry picked from commit 7d89890704ef2ed9265a5dec9cbe47cb1ffc5aaa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2cddc108b174895b3575e9a2640d3cd8afc21ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2cddc108b174895b3575e9a2640d3cd8afc21ab"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
