)]}'
{"id":"openstack%2Fcinder~767478","triplet_id":"openstack%2Fcinder~stable%2Fussuri~I00f1bd46bc8715591016f59c32651fc2c9cddf35","project":"openstack/cinder","branch":"stable/ussuri","topic":"bug/1890591","hashtags":[],"change_id":"I00f1bd46bc8715591016f59c32651fc2c9cddf35","subject":"[SVF]:Reduce slowness by caching pool information","status":"MERGED","created":"2020-12-17 08:30:23.000000000","updated":"2021-03-02 21:26:03.000000000","submitted":"2021-03-02 21:18:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"767478-bug/1890591","cherry_pick_of_change":767477,"cherry_pick_of_patch_set":1,"meta_rev_id":"460b1e3d9eef86d60d741b37393f03e79fc396b4","_number":767478,"virtual_id_number":767478,"owner":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 21:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-02 18:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-02 19:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2021-02-24 06:36:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-02 19:10:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"}],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 08:35:11.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"state":"REVIEWER"},{"updated":"2020-12-17 09:01:25.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 09:03:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 09:37:40.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2020-12-17 10:16:19.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2020-12-17 11:43:35.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 11:50:31.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 12:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 12:49:39.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-18 16:05:43.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-20 12:07:56.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-20 12:56:20.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2020-12-22 09:57:31.000000000","updated_by":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"reviewer":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"state":"CC"},{"updated":"2021-01-27 13:50:43.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2021-01-27 13:51:46.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-02-10 09:37:27.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-03-02 18:51:23.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-02 19:10:36.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"0dd8812419f599f6a990a09833dbba0746d195c5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-17 08:30:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* cinder/tests/unit/volume/drivers/ibm/test_storwize_svc.py\n* cinder/volume/drivers/ibm/storwize_svc/storwize_svc_common.py","accounts_in_message":[],"_revision_number":1},{"id":"2a59d5b853ac91281b2248770d9f5502246a5f8a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-17 08:31:41.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"532ce9d7e9771f286ea46ab3dec0f4b46bc69c86","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-17 09:01:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/78/767478/1/ : RETRY_LIMIT in 2m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c71b651d71a882f5fde59cfc69a288adadfb78f7","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 09:03:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/78/767478/1/check/3par-iscsi-driver-primera/c3be26d/c3be26d.html : FAILURE in 31m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dc76e1f60bf7e37bad5912c909181eb82d35272a","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2020-12-17 09:19:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/78/767478/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/96792d1/96792d1.html : FAILURE in 47m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a4d4c1d2301b423c4ab308246f73adafdf80f956","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-17 09:37:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767478/1/iscsi/ : SUCCESS in 1h 06m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/767478/1/smb/ : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":1},{"id":"13befb897b0a11ef6d7d38dabdc234231fb3a1ee","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-17 10:16:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/78/767478/1/check/dsvm-tempest-cisco-zonemanager-job/3723763/ : FAILURE in 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"bde3b8477352672bf5f47d3aa392ce72878763f1","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-17 11:43:35.000000000","message":"Patch Set 1:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/78/767478/1/check/dsvm-tempest-zadara-driver/51f4961 : FAILURE in 1h 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c627fe8726e57506af6879548955d27d1f45c37","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-17 11:50:31.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-1/ns5_nfs : SUCCESS in 3h 8m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-1/ns5_iscsi : SUCCESS in 3h 3m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"53da5bb9f573bb8f9a04734a48357aa1575bd08d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 12:03:30.000000000","message":"Patch Set 1: Verified-1\n\n(32 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e48f19adc1a4d71b933e70bc2281960 : FAILURE in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d005ed6c063f48be958214cea6d3ca2d : FAILURE in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eed172a0bc0749299010e7cc028b4363 : FAILURE in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df655d15afaa4ff187a3a0befa4d98f5 : FAILURE in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37dff37bda9d4c079d7c5f779eda1f0d : FAILURE in 5m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/3c9c420931bf4030ad57321574e93152 : SUCCESS in 50m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/79e8911f6d584c8488d5cca05bb2e5d9 : FAILURE in 1h 15m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87fc9437faa64e4c85b2fede5f4bcff5 : SUCCESS in 6m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bd3702fba46a40688e3e9bd495295a14 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/483de6a6def14b8b8f008923b08242fc : SUCCESS in 5m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c305997028c04ba0b6a7360db5a53242 : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2edc065932f4418e92d0813368ce0ce3 : SUCCESS in 39m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d90d32f3018b4e4d939bcb35425dacfc : FAILURE in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d915c407a74a484a9bd021021b41f769 : SUCCESS in 1h 19m 22s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6d65f85f13aa409b95550a9ceb2990af : SUCCESS in 1h 44m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e48cd541ed224410b48ef50d198b296b : SUCCESS in 1h 23m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eddb85ee4a7b4d358e45762381c209e4 : SUCCESS in 1h 12m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d1f2bb4ae3fa43e098d37ea3cecd1659 : FAILURE in 1h 17m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee67478e7e7e4cbabd0c319c38e46d52 : SUCCESS in 2h 35m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d3735a3cca74d47acaea509fc667014 : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5e8f99c85ea0e98b5943ac0fac1cd9db0fc8aa96","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-17 12:49:39.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/1/check/tempest-dsvm-full-bionic/e72c853/ : SUCCESS in 1h 59m 42s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/1/check/tempest-dsvm-full-bionic-py3/56c436f/ : SUCCESS in 1h 58m 35s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/1/check/devstack-plugin-ceph-dsvm-bionic-py3/c9b93c1/ : SUCCESS in 1h 56m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8e52f8a141bfe0fc34f3fa6f27c61b14b1b0cf4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-17 16:57:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f2bfda77ccf0523289ab1d3401994fba22d9165f","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-17 18:02:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767478/2/iscsi/ : SUCCESS in 1h 04m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/767478/2/smb/ : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0f621ac65e736875e57ccb5f164a93559cd7feb6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-17 18:38:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/78/767478/2/check/dsvm-tempest-cisco-zonemanager-job/64a71ed/ : FAILURE in 21m 20s","accounts_in_message":[],"_revision_number":2},{"id":"97950a45ee3cc2e3b9c6193595a94623547a69d7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-17 20:51:23.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/2/check/tempest-dsvm-full-bionic/d786271/ : SUCCESS in 2h 29m 19s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/2/check/tempest-dsvm-full-bionic-py3/654c034/ : SUCCESS in 1h 58m 35s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/2/check/devstack-plugin-ceph-dsvm-bionic-py3/981c98e/ : SUCCESS in 2h 20m 25s","accounts_in_message":[],"_revision_number":2},{"id":"843dcad5b4162db33f14bf0e6594b442c76223ec","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-17 22:12:06.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-2/ns5_nfs : SUCCESS in 3h 25m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-2/ns5_iscsi : SUCCESS in 3h 11m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f9ef25604132d05d7095f6049edbd5b66c12c110","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2020-12-18 00:06:42.000000000","message":"Patch Set 2:\n\nSucceeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/78/767478/2/check/dsvm-tempest-zadara-driver/5f1a6f3 : SUCCESS in 1h 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b0620ac74cd802a760524c60c61db49213e53d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 00:34:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/318693b58387468da90fd3503a7e69d1 : FAILURE in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2bfe2ff2a784e5db9033434f244b706 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df0433ffa15844878e3a46fb0e94f658 : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44120fb792af44b581166e3544cfabfe : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc22ae6adcc42948c98bdedebfd41ea : SUCCESS in 29m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/fff90b89ca354bb292f11fe75b5559d1 : SUCCESS in 1h 04m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1e9a86f82c5437facae60cbc2ae2a1e : SUCCESS in 1h 35m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dbdc7536893442ebf8094c2ea5f21db : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/facbc9c360d8473e84d158dd92153106 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b97ebf50fa643049cb91e71e994d593 : SUCCESS in 6m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d90f7935a385411ebf7bdbfbcfa86553 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51f1bdf042fc44829707691bb6fbd5fd : SUCCESS in 34m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/515bb6db542c435593619aa82869e20d : SUCCESS in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d1f16cce3372489589e8945ef58870c7 : SUCCESS in 1h 14m 41s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2e8b6d564d134c97b80f38ef6e7f8b61 : SUCCESS in 1h 55m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc0a527a485b45fb864a7b668a42e860 : SUCCESS in 1h 17m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/edb926828f2049c59f9b39921869c8ed : SUCCESS in 1h 26m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fbf652594e1d4b0baeef7e2bef94f5dc : FAILURE in 1h 05m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e01523222dd43808bab4a6bdf3e2829 : SUCCESS in 2h 07m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e26b4337a8d475faa7acbd3eb2fc1e2 : SUCCESS in 59m 57s","accounts_in_message":[],"_revision_number":2},{"id":"0f0d73e8f25740857e8cf4cf7522f1f6e4a8dfb3","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-12-18 05:41:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/78/767478/2/ : RETRY_LIMIT in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"732f7ba69f87513eef981c5ee2af63fc7739f308","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2020-12-18 16:05:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/78/767478/2/20201218222545/NEC-FC : SUCCESS in 2h 19m 45s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/78/767478/2/20201218222545/NEC-ISCSI : SUCCESS in 2h 39m 56s","accounts_in_message":[],"_revision_number":2},{"id":"39db3825fd0e1c44da97cb2b141eef74d54707b0","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-20 08:56:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4bf1302f1ad301601dd544511f2fc4c3819241d1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-20 09:23:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/78/767478/2/check/dsvm-tempest-cisco-zonemanager-job/a613ca7/ : FAILURE in 21m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ba6e764c62d93777824424ea71500d5bb46e4780","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2020-12-20 10:00:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767478/2/iscsi/ : SUCCESS in 1h 03m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/767478/2/smb/ : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":2},{"id":"06824a26442daeadfffab88aeb9d7c6796b9c6ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 11:09:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db459dcaefe44230bafb61cd76f0698c : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4732d68bc5494b6e8ee9832bf2e94ebd : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0f5c3f8c8774f0c820954e9a227eacd : SUCCESS in 11m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6cdc0b24e614cc7abd114dcb3395d6f : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09727eab7eba4013a78b82d6aa7c71a2 : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/c71dd7ef3d774f8ca24470583fb24f45 : SUCCESS in 1h 04m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a75c87c1fda8400695c883585585efdb : SUCCESS in 1h 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53b373267e814482bf4cd20e9501dc23 : SUCCESS in 7m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee26c20343d6406da74186764672caa1 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/deee177fe3764ac586104cdcf6e9b9df : SUCCESS in 8m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/794e0b5f4cfa4c2fb4694fcdbf0ca8b0 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/07110f0194354df78c7d67325f602054 : SUCCESS in 41m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88d42e4cac474b58b93957fd8c6fa6c9 : SUCCESS in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2183b4658ba42faa149b37df8ba860f : SUCCESS in 1h 09m 51s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/756444a58679458a93901a99e09b701b : SUCCESS in 1h 36m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/04316ee5facd4f28a797967b5a1d5edf : SUCCESS in 1h 26m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d01a18f2f9d74c5ca1384fcdb2bc734a : SUCCESS in 1h 18m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/287c278ea0b246869f5c9e0c36c196f6 : FAILURE in 1h 05m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03af04a547e54ebbb89ed1bf9c7da15d : SUCCESS in 1h 53m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cd691bb84084177953ca06c0610f82c : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b915120f1312b3e6103fa997737a4c7b10132574","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-20 11:40:25.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/2/check/tempest-dsvm-full-bionic/bbf03f5/ : SUCCESS in 1h 44m 25s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/2/check/tempest-dsvm-full-bionic-py3/b90197a/ : SUCCESS in 1h 51m 43s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/2/check/devstack-plugin-ceph-dsvm-bionic-py3/34c9c54/ : TIMED_OUT in 2h 39m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9080cedd3433f0a20fa3f411720082e87caad30b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-20 12:07:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/78/767478/2/upstream-check/cinder-cDOT-iSCSI/ec0c383/ : SUCCESS in 37m 25s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/78/767478/2/upstream-check/cinder-cDOT-iSCSI-IPv6/b7fef5a/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/78/767478/2/upstream-check/cinder-cDOT-NFS/5d1a258/ : FAILURE in 34m 11s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/78/767478/2/upstream-check/cinder-cDOT-NFS-IPv6/b0e4ed1/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/78/767478/2/upstream-check/cinder-cDOT-FCP/10adaed/ : SUCCESS in 57m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b1d503cd47062e7641a227bc20434b56b5a6b19","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-12-20 12:56:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/78/767478/2/upstream-solidfire-check/cinder-solidfire-iSCSI/931025d/ : FAILURE in 1h 45m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"4baf43d10ce0eb4eaf544777a0fe72d6293dea3e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-21 06:00:39.000000000","message":"Patch Set 2:\n\ncisco-cinder-ci","accounts_in_message":[],"_revision_number":2},{"id":"fa4bee12905ffa3b5e9c796152b0bc8ad0571821","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-12-21 06:27:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/78/767478/2/check/dsvm-tempest-cisco-zonemanager-job/abcb642/ : FAILURE in 21m 57s","accounts_in_message":[],"_revision_number":2},{"id":"086ad000660b551432f928212bcba73c153ccf54","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-22 04:56:28.000000000","message":"Patch Set 2:\n\nrun-IBM Storage CI","accounts_in_message":[],"_revision_number":2},{"id":"166db0171eeaaf5c505e3b396a54fc15c387fbc2","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-22 08:00:21.000000000","message":"Patch Set 2:\n\nrun-IBM Storage CI","accounts_in_message":[],"_revision_number":2},{"id":"e927908d2f417418c9e1b7024847d856491705e6","author":{"_account_id":18883,"name":"IBM Storage CI","email":"IBM_CI@il.ibm.com","username":"ibm_storage_ci"},"date":"2020-12-22 09:57:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).  Leave a comment with \u0027run-IBM Storage CI\u0027 to trigger a recheck. please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI\n\n- ds8k-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ds8k-tempest-dsvm-full-fc/1992/index.html : SUCCESS in 1h 13m 14s\n- storwize-tempest-dsvm-full-iscsi https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/2138/index.html : SUCCESS in 1h 55m 06s\n- storwize-tempest-dsvm-full-fc https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/2127/index.html : SUCCESS in 1h 27m 03s","accounts_in_message":[],"_revision_number":2},{"id":"052dbbb4965b0eb1e17e828df92730fec25eb510","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-27 13:51:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nVictoria patch needs to be merged first.\n-1 for the inline comment","accounts_in_message":[],"_revision_number":2},{"id":"015edc1bd2213843ccd96bbe201d838dde43c254","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2021-02-06 16:48:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17989e60ad094607677728af11756d988926ea81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-10 09:06:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e20fef9046ce95db1289c69a6f11442e9054438d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-10 09:11:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"036e1aeb3c598e4b88b4394e32b1b004a6a8dbe6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-10 09:36:19.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/78/767478/3/check/dsvm-tempest-cisco-zonemanager-job/47a786d/ : FAILURE in 24m 27s","accounts_in_message":[],"_revision_number":3},{"id":"17857cec448582520dcd8cad4bcd984150feb1da","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-10 09:37:27.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66:9090/78/767478/3/check/dsvm-tempest-TOYOU-ACS5000-driver/12b91da : FAILURE in 29m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ca261a910b365a6efdd19665b1281da9771f8946","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-10 09:58:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/78/767478/3/check/iscsi-driver-primera/e309f23/e309f23.html : FAILURE in 50m 24s","accounts_in_message":[],"_revision_number":3},{"id":"f434579d6c693ce75c0b26677a035045c5ceddb7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-10 10:18:15.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/3/check/tempest-dsvm-full-bionic/8912e56/ : FAILURE in 51m 32s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/3/check/tempest-dsvm-full-bionic-py3/063563d/ : FAILURE in 50m 58s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/3/check/devstack-plugin-ceph-dsvm-bionic-py3/f08ba62/ : FAILURE in 1h 05m 51s","accounts_in_message":[],"_revision_number":3},{"id":"b2231a024b23e8851d78531436ef1075d668534e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-10 10:23:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/78/767478/3/check/dsvm-tempest-zadara-driver/dcf745d : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 16m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09bd33fc14bce95108dec80816d6a1d03218d79c","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-10 10:37:01.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767478/3/iscsi/ : FAILURE in 1h 29m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/767478/3/smb/ : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a8356bdd200934475209dfdbd83e57f2613ba887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 11:18:43.000000000","message":"Patch Set 3: Verified-1\n\n(30 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a01f98557075494893f0eb1bc39673fc : FAILURE in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e3253db09b949f9a8850182cb959122 : FAILURE in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5203cf22e394064859b6e43307ef1f9 : FAILURE in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1216496f9dae4175bad67c51def56c52 : FAILURE in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d752268a27490791071a7d63ce0fd1 : FAILURE in 5m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/a81ee88ca97547abb37fdafdcd93dba7 : SUCCESS in 48m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9d703583ab34c9189569d1f5847e09a : SUCCESS in 1h 15m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d6ba915e5cc47bcb996181161a8d063 : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/efd90f770d65488d9fc84082d07e07b0 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb2737a9401f40e79963e90a72fa5b7d : SUCCESS in 5m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/da9404fee085465eb2c0af8afc337411 : SUCCESS in 6m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/22f3c17656834593ad9ff18fcddad7fe : SUCCESS in 34m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/57a7442ca4dd4bf58ba1d4a8dd28218d : FAILURE in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/09b1b118d214444094a3aa0b1a76f6c8 : SUCCESS in 58m 08s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/50907655942e4049a70b3439873fb002 : FAILURE in 1h 24m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/08dd9a88ea13434fa5818d41c3edb1ab : SUCCESS in 1h 14m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1dbc402c1fb04d56a44f50534d43036a : SUCCESS in 1h 26m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2b87b706864040bfbe132c55dd28083e : FAILURE in 1h 15m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e268a613eeca4d3f9d9eea4f5a343c66 : SUCCESS in 2h 06m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42cc5cbd409a41408a60957ee42e77de : SUCCESS in 53m 04s","accounts_in_message":[],"_revision_number":3},{"id":"3bb48d4b6864eced98e939b1d7a31aff50eb7f14","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-10 11:41:40.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-3/ns5_nfs : FAILURE in 2h 31m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-3/ns5_iscsi : FAILURE in 2h 34m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"4abd764020c61cc26a64a830cf69b77e712d1bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-10 12:31:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"090f233fa9b73bd6dd80d02d26a4f63e0168f000","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-10 12:56:34.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66:9090/78/767478/4/check/dsvm-tempest-TOYOU-ACS5000-driver/54be4cf : FAILURE in 18m 46s","accounts_in_message":[],"_revision_number":4},{"id":"dfb9a989ae6d34902bda6db0fd907512146673db","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-10 13:28:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/78/767478/4/check/iscsi-driver-primera/8a96669/8a96669.html : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":4},{"id":"5b69b8f33d05693406abdbff9b70d995a6510d09","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-10 13:35:01.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/767478/4/iscsi/ : SUCCESS in 1h 03m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/767478/4/smb/ : FAILURE in 44m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f98107886d0322c1b5eb53e45ccc60eb14512f6a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-10 13:49:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/78/767478/4/check/dsvm-tempest-zadara-driver/11dfcb2 : SUCCESS in 1h 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"43499c2161e42f37f2c973f52910054490b14e40","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-10 13:49:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/78/767478/4/check/dsvm-tempest-cisco-zonemanager-job/1cf2dee/ : FAILURE in 24m 25s","accounts_in_message":[],"_revision_number":4},{"id":"3359efff7326bf0dc2702e615b6caf27cea3f336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 14:39:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3115d32421f74c9495d3e202203c40ec : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9620489ab70461da2237dd36c57ea27 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5751b4e737b14fb99226a8122868695c : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9328193552fd48d18a03d4ebaebff85a : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1dbd8afc6aa4b79a2cdda4730151479 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/e1906cd753a84aeaa01a56921dba6164 : SUCCESS in 1h 06m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd5e6fa73cfd4e0baba8ec4304148452 : SUCCESS in 1h 08m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2f511ea36ec4b64a56d4121aa8c0083 : SUCCESS in 6m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/def410ef9b0145cebe4c43b8624baf6d : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21d6232e981c41f694da2afa5ae122fb : SUCCESS in 7m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/60aabfd4e8804120b24099f87f3f51eb : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d0f5e68fecb74fd199bfdcbcf53804e2 : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7940d950aacf4e2ab4e3da93d5fe9570 : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fde586438a4f45cd8ad5f6e1d9797812 : SUCCESS in 55m 45s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec6ba1d4c0c74c06b7f2330179913e8a : SUCCESS in 1h 31m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f79761b42a1141cda628eea45fe6385b : SUCCESS in 1h 19m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df5f967160fe42bbbd2abdbac4b51dd2 : SUCCESS in 1h 22m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d8ad3407194347fd970f363f0fb70432 : FAILURE in 1h 12m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba5ee88058fd4723aa44489202be52d1 : SUCCESS in 2h 01m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d7360676b5540959ee952ce9305f75d : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":4},{"id":"b7db28aaadd87fa43174a8df84ea24d775549858","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-10 15:08:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/78/767478/4/upstream-solidfire-check/cinder-solidfire-iSCSI/5617a57/ : FAILURE in 27m 51s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"308e4848df5c75a6026ce9bf3475fa65e6cebfac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-10 15:38:04.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/4/check/tempest-dsvm-full-bionic/5917bae/ : SUCCESS in 2h 06m 53s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/4/check/tempest-dsvm-full-bionic-py3/e65c4ea/ : SUCCESS in 1h 46m 35s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/767478/4/check/devstack-plugin-ceph-dsvm-bionic-py3/5518f1f/ : TIMED_OUT in 2h 38m 12s","accounts_in_message":[],"_revision_number":4},{"id":"b2292a84ce81b0a637760fa3c94b89f0f6270654","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-10 15:53:18.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-4/ns5_nfs : FAILURE in 3h 21m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-4/ns5_iscsi : FAILURE in 2h 57m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"a9b13d4c4c1bc77053eb123a130136c724d4f92b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-10 16:15:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/78/767478/4/upstream-check/cinder-cDOT-iSCSI/3663f6d/ : FAILURE in 20m 01s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/78/767478/4/upstream-check/cinder-cDOT-iSCSI-IPv6/7067b81/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/78/767478/4/upstream-check/cinder-cDOT-NFS/0fff65b/ : FAILURE in 1h 12m 49s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/78/767478/4/upstream-check/cinder-cDOT-NFS-IPv6/107e3fd/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/78/767478/4/upstream-check/cinder-cDOT-FCP/0ee6498/ : FAILURE in 54m 20s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"dbfd33959c83054fc3fc1480c4c4ed9b609594a2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-10 20:57:08.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-4/ns5_nfs : FAILURE in 3h 41m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-4/ns5_iscsi : FAILURE in 3h 35m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"337b1bad07c1f86c69f4fce424356615332182f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-11 01:35:19.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-4/ns5_nfs : FAILURE in 4h 40m 34s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-78-767478-4/ns5_iscsi : FAILURE in 4h 26m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"dde122eb4b4e9b5045fc294005ba057e0a70893a","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-11 03:58:40.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/78/767478/4/20210211103946/NEC-ISCSI : SUCCESS in 2h 11m 40s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/78/767478/4/20210211103946/NEC-FC : SUCCESS in 2h 18m 51s","accounts_in_message":[],"_revision_number":4},{"id":"5ce1507e257212b68fb755eb0d1c357050f54227","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-24 06:36:42.000000000","message":"Patch Set 4: Code-Review+1\n\nMerged in victoria, looks good to backport.","accounts_in_message":[],"_revision_number":4},{"id":"09dbccb01e96d53ac690c69d40ca90beebcfefc4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-02 18:51:23.000000000","message":"Patch Set 4: Code-Review+2\n\nSince we thought it was good for Victoria, no reason to keep it out of Ussuri (and for the same reasons).","accounts_in_message":[],"_revision_number":4},{"id":"b2e331769b2e9cc95200c83740d0fb92df3c3acb","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-02 19:10:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLooks like a good backport.  Thanks!","accounts_in_message":[],"_revision_number":4},{"id":"dc638cae6db941916db49786ca8c7298fc725348","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 19:14:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"358a78b3e6172f869848f3188a2f408f842ca8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:17:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47f0d2e034f14626ac5e8de67b6a264a : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f5347af8c5412691d13824037de7b5 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83492c7887a14a63a3f85ca773deca98 : SUCCESS in 11m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d19454aec4cc49908548ea07c59052f1 : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602c2176ed0647b889e62089a7e3e32c : SUCCESS in 15m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/e507dab26a4b4b1b9c98054c92eaf1b5 : SUCCESS in 1h 05m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aca6d4a22b334a42a84da0e109f708f4 : SUCCESS in 1h 06m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c30aa0b52e834f10a4a95c627ee8c400 : SUCCESS in 7m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/658f3d3588404153963b1733490b41be : SUCCESS in 1h 14m 49s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e22e80950f1b477a9957d3d506163ebd : SUCCESS in 1h 08m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce15d30c4ee74d63bef1f8017e46508b : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":4},{"id":"3696c467adea263f505c6e9a7f2c5b562a3f16a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:18:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"460b1e3d9eef86d60d741b37393f03e79fc396b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:26:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cec277e48464faf88e428e67b35e8db : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fea538ee2594e75ba6740dd3988e3f3 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8426a5bdc0c2becdab87920740bee8e1bf983992","revisions":{"1b36a23c546a4d1702ecfb9bf1379014668665b3":{"kind":"REWORK","_number":1,"created":"2020-12-17 08:30:23.000000000","uploader":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"ref":"refs/changes/78/767478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/767478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/767478/1"}}},"commit":{"parents":[{"commit":"2f0d236440b6816ae2e50ae10192777fcf4117dd","subject":"Merge \"Fix volume rekey during clone\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f0d236440b6816ae2e50ae10192777fcf4117dd"}]}],"author":{"name":"GirishChilukuri","email":"girish.chilukuri@ibm.com","date":"2020-09-25 20:31:25.000000000","tz":0},"committer":{"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","date":"2020-12-17 08:30:23.000000000","tz":0},"subject":"[SVF]:Reduce slowness by caching pool information","message":"[SVF]:Reduce slowness by caching pool information\n\n[Spectrum Virtualize Family] Data_reduction is a pool specific property,\nthis property should be saved for a specific pool during initialisation.\n\nWith this, calls to get_pool_attrs can be reduced everytime create_vdisk\nis called, in order to check if pool is a data_reduction_pool.\n\ncloses bug: #1890591\n\nChange-Id: I00f1bd46bc8715591016f59c32651fc2c9cddf35\n(cherry picked from commit 7d89890704ef2ed9265a5dec9cbe47cb1ffc5aaa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b36a23c546a4d1702ecfb9bf1379014668665b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b36a23c546a4d1702ecfb9bf1379014668665b3"}]},"branch":"refs/heads/stable/ussuri"},"74d244893bc96cfb6c3352f3b05e646bb26d6177":{"kind":"REWORK","_number":2,"created":"2020-12-17 16:57:11.000000000","uploader":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"ref":"refs/changes/78/767478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/767478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/767478/2"}}},"commit":{"parents":[{"commit":"2f0d236440b6816ae2e50ae10192777fcf4117dd","subject":"Merge \"Fix volume rekey during clone\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f0d236440b6816ae2e50ae10192777fcf4117dd"}]}],"author":{"name":"GirishChilukuri","email":"girish.chilukuri@ibm.com","date":"2020-09-25 20:31:25.000000000","tz":0},"committer":{"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","date":"2020-12-17 16:52:01.000000000","tz":0},"subject":"[SVF]:Reduce slowness by caching pool information","message":"[SVF]:Reduce slowness by caching pool information\n\n[Spectrum Virtualize Family] Data_reduction is a pool specific property,\nthis property should be saved for a specific pool during initialisation.\n\nWith this, calls to get_pool_attrs can be reduced everytime create_vdisk\nis called, in order to check if pool is a data_reduction_pool.\n\ncloses bug: #1890591\n\nChange-Id: I00f1bd46bc8715591016f59c32651fc2c9cddf35\n(cherry picked from commit 7d89890704ef2ed9265a5dec9cbe47cb1ffc5aaa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74d244893bc96cfb6c3352f3b05e646bb26d6177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74d244893bc96cfb6c3352f3b05e646bb26d6177"}]},"branch":"refs/heads/stable/ussuri"},"e6f593a44fcc340256bb6b631933dd08312c728e":{"kind":"REWORK","_number":3,"created":"2021-02-10 09:06:35.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/78/767478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/767478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/767478/3"}}},"commit":{"parents":[{"commit":"4f40f059f81ac27c92e4238f78ba649b69b09dd0","subject":"RBD: Retry delete if VolumeIsBusy in _copy_image_to_volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f40f059f81ac27c92e4238f78ba649b69b09dd0"}]}],"author":{"name":"GirishChilukuri","email":"girish.chilukuri@ibm.com","date":"2020-09-25 20:31:25.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-02-10 09:06:35.000000000","tz":0},"subject":"[SVF]:Reduce slowness by caching pool information","message":"[SVF]:Reduce slowness by caching pool information\n\n[Spectrum Virtualize Family] Data_reduction is a pool specific property,\nthis property should be saved for a specific pool during initialisation.\n\nWith this, calls to get_pool_attrs can be reduced everytime create_vdisk\nis called, in order to check if pool is a data_reduction_pool.\n\ncloses bug: #1890591\n\nChange-Id: I00f1bd46bc8715591016f59c32651fc2c9cddf35\n(cherry picked from commit 7d89890704ef2ed9265a5dec9cbe47cb1ffc5aaa)\n(cherry picked from commit e2cddc108b174895b3575e9a2640d3cd8afc21ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6f593a44fcc340256bb6b631933dd08312c728e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6f593a44fcc340256bb6b631933dd08312c728e"}]},"branch":"refs/heads/stable/ussuri"},"8426a5bdc0c2becdab87920740bee8e1bf983992":{"kind":"REWORK","_number":4,"created":"2021-02-10 12:31:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/78/767478/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/767478/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/767478/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/767478/4"}}},"commit":{"parents":[{"commit":"4f40f059f81ac27c92e4238f78ba649b69b09dd0","subject":"RBD: Retry delete if VolumeIsBusy in _copy_image_to_volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f40f059f81ac27c92e4238f78ba649b69b09dd0"}]}],"author":{"name":"GirishChilukuri","email":"girish.chilukuri@ibm.com","date":"2020-09-25 20:31:25.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-02-10 12:31:18.000000000","tz":-300},"subject":"[SVF]:Reduce slowness by caching pool information","message":"[SVF]:Reduce slowness by caching pool information\n\n[Spectrum Virtualize Family] Data_reduction is a pool specific property,\nthis property should be saved for a specific pool during initialisation.\n\nWith this, calls to get_pool_attrs can be reduced everytime create_vdisk\nis called, in order to check if pool is a data_reduction_pool.\n\ncloses bug: #1890591\n\nChange-Id: I00f1bd46bc8715591016f59c32651fc2c9cddf35\n(cherry picked from commit 7d89890704ef2ed9265a5dec9cbe47cb1ffc5aaa)\n(cherry picked from commit e2cddc108b174895b3575e9a2640d3cd8afc21ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8426a5bdc0c2becdab87920740bee8e1bf983992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8426a5bdc0c2becdab87920740bee8e1bf983992"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
