)]}'
{"id":"openstack%2Fcinder~768837","triplet_id":"openstack%2Fcinder~master~I86bc7f14d7d9006e93b33f9453b5727c9bb26c8f","project":"openstack/cinder","branch":"master","topic":"bp/six-removal","hashtags":[],"change_id":"I86bc7f14d7d9006e93b33f9453b5727c9bb26c8f","subject":"Remove six from quota.py","status":"MERGED","created":"2021-01-02 02:11:58.000000000","updated":"2021-04-21 04:20:56.000000000","submitted":"2021-04-20 07:49:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"768837-bp/six-removal","meta_rev_id":"445285eccfb47c66dbfc79af3e56ad4ac6f48929","_number":768837,"virtual_id_number":768837,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-20 07:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-24 10:26:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2021-03-19 20:29:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":2,"date":"2021-04-19 16:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-19 15:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2021-03-22 03:25:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":1,"date":"2021-04-19 16:56:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32925,"name":"lisa","email":"lisa.qiao@dell.com"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-02 02:12:11.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 02:53:29.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-01-02 02:54:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 03:35:22.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-01-02 03:44:51.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 03:55:10.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 04:01:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-02 04:20:01.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 04:23:48.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 04:27:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-02 05:06:22.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 07:10:00.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 05:43:53.000000000","updated_by":{"_account_id":32925,"name":"lisa","email":"lisa.qiao@dell.com"},"reviewer":{"_account_id":32925,"name":"lisa","email":"lisa.qiao@dell.com"},"state":"CC"},{"updated":"2021-01-06 06:31:26.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-01-06 08:30:26.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-01-06 08:39:30.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-01-06 09:14:42.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 09:25:54.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-01-06 09:57:39.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 15:37:37.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 08:25:11.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"CC"},{"updated":"2021-02-15 08:57:38.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-15 09:12:26.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-15 11:43:18.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 14:10:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-19 15:21:19.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2021-02-19 16:30:06.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-02-20 06:40:48.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-20 22:31:38.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-02-21 11:41:30.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-03-19 15:20:38.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2021-03-19 15:20:52.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-19 20:29:04.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2021-03-22 03:25:42.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2021-03-24 10:26:02.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-04-19 16:56:36.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-04-20 05:56:49.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"}],"messages":[{"id":"0effd5b0a380164afcf273708f501c4f1a34db88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-01-02 02:11:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6566c3ea9af348ef05cac868bec415f4d44c0f44","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-02 02:12:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f11df249980d0f5940ae444d2d74c3c18e4baf56","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-02 02:53:29.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768837/1/iscsi/ : FAILURE in 40m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/768837/1/smb/ : FAILURE in 35m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0234121e6084dc8ee13ffb8e4318728361db5023","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-02 02:54:41.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/1/check-not-vote/ext-cinder-zuul/eeade37 : SUCCESS in 41m 59s","accounts_in_message":[],"_revision_number":1},{"id":"459cf6c2ceadec42c84fe6b4dc079363f964b18a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-02 03:35:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/1/check/dsvm-tempest-cisco-zonemanager-job/3c35f4a/ : SUCCESS in 1h 18m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b8284873945916545c7c1c068eb594eac21759c3","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-02 03:44:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/37/768837/1/ : FAILURE in 1h 24m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ee8b27322540ad9e5c8d58b5f94b1de708afeb40","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-02 03:55:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/37/768837/1/check/DellEMC_XtremIO_FC/06cb03d/DellEMC_XtremIO_FC/408 : SUCCESS in 1h 42m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/37/768837/1/check/DellEMC_XtremIO_iSCSI/d189f19/DellEMC_XtremIO_iSCSI/394 : SUCCESS in 1h 35m 19s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"27ff66bffa4ca8ec94e93b58170076b67db0105c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-02 04:01:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/37/768837/1/check/DellEMC_Unity_FC/ba1029f/DellEMC_Unity_FC/394 : FAILURE in 1h 48m 51s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/37/768837/1/check/DellEMC_Unity_iSCSI/35f3fa9/DellEMC_Unity_iSCSI/369 : FAILURE in 1h 22m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e05266fdb251c245bccbd6814862ddded7947649","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-02 04:20:01.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b06ab47e7b0b8fd222de5efb9cf34321a918e230","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-02 04:23:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/1/check/tempest-dsvm-full-focal-py3/6a3e1ed/ : SUCCESS in 2h 03m 43s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/1/check/devstack-plugin-ceph-dsvm-focal-py3/624399e/ : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4380ce2bd7c571c61a3d000f493d35292e45a55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 04:27:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/536f0541572d40da8412c7b3c8af881f : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f83217185340e89fa6d571f6cfb1db : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa042d6449084cafb2b66402ab165a5a : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8468897bedca45b4bb92cdb1e2e3c552 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5afc724e146486992413904c128281f : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04c8fd3ddbb4fda82fe0d2d2e169e03 : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/06181569a58f4386a38959141062da12 : SUCCESS in 57m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00582cbc12bc4932abf6e06e750ca18a : SUCCESS in 1h 14m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6cfccc7ed06b47869d8bb5c082bdacbc : SUCCESS in 11m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc7124de11d0492795b8d0d88b095f31 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22e2c3d45d4d47b1a73b210ed1e7ac55 : SUCCESS in 9m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/95386d7a05d546b3961b62cdbf48c587 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7ac54ea66d3e4cf482050ce1ed962cd5 : SUCCESS in 42m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1762459b4a2442e98fa744c5fa6071da : FAILURE in 4m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/21304f4ff4a248108ec032064eae9ea4 : SUCCESS in 1h 16m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6941ef5e597e4adcb49de574e020dd91 : FAILURE in 1h 16m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a2c7c509e3643569379b560e2d73406 : FAILURE in 1h 25m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c4f87dbe78d47a2ac8beeaf3818a1a2 : SUCCESS in 1h 10m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7cd5924e3e364a5cae7f21ef3118ef4f : SUCCESS in 1h 12m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc2897a9fd494a92b67652b0b6757cb8 : SUCCESS in 1h 31m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9917776eaee4cebab0199ce28b916a8 : SUCCESS in 2h 10m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9971026b9d6444938656f8c7bba1383f : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b99ec7ae8d771dcb2d5c518ba2156836cad2f84b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-02 05:06:22.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-37-768837-1/ns5_nfs : SUCCESS in 2h 51m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-37-768837-1/ns5_iscsi : SUCCESS in 2h 53m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7b45a5a049c95627dfd3a10f1babf9761a91d7ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-02 05:36:09.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4a03d16ae72f29d83d94c5ce3ab4c1e9fcdb2424","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-02 07:10:00.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/37/768837/1/check/DellEMC_PowerMAX_FC/52c0111/DellEMC_PowerMAX_FC/329 : SUCCESS in 3h 59m 36s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/37/768837/1/check/DellEMC_PowerMAX_iSCSI/450cf03/DellEMC_PowerMAX_iSCSI/302 : SUCCESS in 4h 46m 16s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"5a9a7de8548e92aaeebb40cb8bc31d65ce038fb5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-02 08:25:10.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7f3c288db95e0602a76514323c6a7b1f4a59f81a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-03 15:31:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/37/768837/1/ : RETRY_LIMIT in 24s","accounts_in_message":[],"_revision_number":1},{"id":"1c7c3daf64741f8724c87e767472f6f9fd99c875","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-03 18:54:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/37/768837/1/ : RETRY_LIMIT in 24s","accounts_in_message":[],"_revision_number":1},{"id":"6c3821b8c3818d30c1a04adf06e9198babe6d5d4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-03 22:53:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/37/768837/1/ : FAILURE in 2h 03m 01s","accounts_in_message":[],"_revision_number":1},{"id":"28d8a00910740353b1b72542c18c20a6cd872690","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-01-05 07:34:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"591ac39134b31d2efa3bb69990a1ef82519d0300","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-05 07:34:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff0b0e4f25b4f8b81068ebcd09a91229c13493e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-05 08:18:12.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/1/check-not-vote/ext-cinder-zuul/2b5e7ec : SUCCESS in 43m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b91c6426aaa24da76b4e51ad9b2b6f05a5ba793b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-05 08:53:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/1/check/dsvm-tempest-cisco-zonemanager-job/78e5d2e/ : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1575a434cc58531bfedd340b5d4db1ac4bd0e3f0","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-05 09:01:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768837/1/iscsi/ : FAILURE in 1h 27m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/768837/1/smb/ : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":1},{"id":"25c0f6f79cfed79c672f9996758da97a5c6b0273","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-05 10:00:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"27fce813f138f99b50578d67e4b98e482ed392e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 10:32:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d14509bcf5d4b018cccd591ac233187 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e33117f97034f329912e1e79dab25da : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20980a9db780403eb0927ac6979f261f : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3db7b873de3484d9d8f27a650246ddc : SUCCESS in 9m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abb111f9b9a24b808976d35ea1085237 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48775c55882942de9dc8f00afdeba62e : SUCCESS in 18m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/448fc154ed984b4d97f4038372817783 : SUCCESS in 1h 10m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d7285343a944d06bf53c6cce53aac42 : TIMED_OUT in 2h 08m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dd53122ca842464e8f1a4e43aee0cb21 : SUCCESS in 17m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/50745757cd344c29b342ac8d7414b076 : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c8b168f3eaa04fa288872f6ff4aa6205 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/62e70974145e4c118efcad912d4d0589 : SUCCESS in 6m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c1062f2af707424d84edd92812a33cb5 : SUCCESS in 36m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e73ad6765bf4490880407ca76fc21700 : FAILURE in 8m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7086b6da37984f6ea63221bdbbd6f8f2 : SUCCESS in 54m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/120ad60a095b4fc9a1ceaa46f1e51020 : FAILURE in 1h 36m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d9689b4df28a45cb8188e7ff95ab3a48 : SUCCESS in 1h 55m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc81185686e34b4b96fa4724b94bc82a : SUCCESS in 2h 03m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e1cbe38768ee42ecbfa6f9be36b31ffe : FAILURE in 1h 19m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/004bf285dcd841b5b6eb3b5f35cec5f6 : SUCCESS in 1h 20m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/007ae46e6a1145b29ba44c536d873333 : FAILURE in 2h 56m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/628aef144ea64acb82fe6c6f0b0f00d0 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b9d8cbc73baab67d3c29153c421f31077bcb6ab2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-05 10:35:47.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/1/check/tempest-dsvm-full-focal-py3/128cc97/ : SUCCESS in 2h 46m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/1/check/devstack-plugin-ceph-dsvm-focal-py3/8f8d066/ : SUCCESS in 1h 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"121b31f11708272c9d4a2fe0370b4a631527a7b4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-05 12:48:21.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8e0b04e72f8cd1d9752e7c0d7530f84da2054400","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-01-06 03:20:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1a0887761b5d41da052f730fabfbc88853006c0","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-06 03:21:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37bd2d4f2a06b690ecec4fbfdcaa8645f29e5dd1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-06 03:26:53.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/1/check-not-vote/ext-cinder-zuul/1532a69 : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3920d8fe0cb8f1769d9cb6c475698f0bdbefd728","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-06 04:27:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768837/1/iscsi/ : SUCCESS in 1h 06m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/768837/1/smb/ : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6d4aa285461649e51fb62d4184fb178a039b8601","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-06 04:42:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/1/check/dsvm-tempest-cisco-zonemanager-job/7d40817/ : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3cfc4ec60b2762f24b94cf39253c57e92e1248fb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-06 05:01:11.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/1/check/tempest-dsvm-full-focal-py3/592fd29/ : SUCCESS in 1h 24m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/1/check/devstack-plugin-ceph-dsvm-focal-py3/c28ed35/ : SUCCESS in 1h 35m 38s","accounts_in_message":[],"_revision_number":1},{"id":"524d2a14e0cae6b08e1d9edf950856f59fe57dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 05:33:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d47d64bc04334256a77cfa3566a30c06 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/886c4ad2499a48e887fff7d842f29ab9 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49887d2d37f9444096e1181642b4f368 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b450a58c982543d995b707d00031a5ca : SUCCESS in 9m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67ad5f9a34db41498fcb13be03751273 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e34ce4953c5485d980c981f727b5afb : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/978bc4535b9a44fcb42a6b55d8f551b8 : SUCCESS in 1h 03m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/776a115c01404ed1aa3edf8890c60f22 : SUCCESS in 1h 13m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90d72200b52348ee92ee305a0eadf395 : SUCCESS in 12m 05s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93f69d6419a842749a83abda74fb0044 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acc858fd492e4c18a095bf446f71a685 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7fdfca2a3f4346339fb5e17cc81788ce : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/535bd285ea6a4d7189859b0644daee68 : SUCCESS in 44m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c6b3ac8aa6aa46c0b97e6d895a83ed67 : SUCCESS in 4m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d81f20c394c74bc5a459c13c07cc9fe3 : SUCCESS in 1h 13m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/841cef85dccc4af5848a22e01d3235d7 : FAILURE in 1h 20m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/97c03f06a5b74e0ea37e6742960fe3b4 : SUCCESS in 1h 21m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b77c031253e9491f89cdda8e24f4cc4e : SUCCESS in 1h 12m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7cee1a834c51439ebc666604d03cd719 : SUCCESS in 1h 28m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a8e84e9203df4da9945646aca271f029 : SUCCESS in 1h 22m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/267f9da0e513455a96e04ab418e93e8b : SUCCESS in 2h 06m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b35ba5433b143dba9c46158ad4fa07d : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9849e1b7660ca1e7fce30e3edaf4afdb52dfbeff","author":{"_account_id":32925,"name":"lisa","email":"lisa.qiao@dell.com"},"date":"2021-01-06 05:43:53.000000000","message":"Patch Set 1:\n\nrun-DellEMC XtremIO CI","accounts_in_message":[],"_revision_number":1},{"id":"aeb47a7e5552eee3682c50fe8a9a00523916fa13","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-06 06:31:26.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/37/768837/1/check/DellEMC_PowerStore_iSCSI/5147c80/DellEMC_PowerStore_iSCSI/235 : SUCCESS in 57m 41s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/37/768837/1/check/DellEMC_PowerStore_FC/aaf38c7/DellEMC_PowerStore_FC/229 : SUCCESS in 54m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"3fce7784ba12ebb2f6c94a27b0930554cfb80439","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-06 07:17:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/37/768837/1/check/DellEMC_XtremIO_FC/dfc0a18/DellEMC_XtremIO_FC/453 : FAILURE in 15m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/37/768837/1/check/DellEMC_XtremIO_iSCSI/8bdc881/DellEMC_XtremIO_iSCSI/442 : SUCCESS in 1h 33m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"c76427d2925ed56386432c275aeb667c446c2b8f","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-06 08:30:26.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/768837-1/Seagate-iSCSI : SUCCESS in 47:18.77\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/768837-1/Seagate-FC : SUCCESS in 41:55.45","accounts_in_message":[],"_revision_number":1},{"id":"4959359bf6048e9b2c468d23701b8d2ec5bc3ebe","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-06 08:39:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/37/768837/1/upstream-solidfire-check/cinder-solidfire-iSCSI/38103df/ : SUCCESS in 53m 07s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"d7ab75ec98dd8e3d6efa38de01344009d951c653","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-06 09:14:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/37/768837/1/upstream-check/cinder-cDOT-iSCSI/b690aca/ : SUCCESS in 37m 38s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/37/768837/1/upstream-check/cinder-cDOT-iSCSI-IPv6/f3e6a39/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/37/768837/1/upstream-check/cinder-cDOT-NFS/7395f16/ : SUCCESS in 35m 43s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/37/768837/1/upstream-check/cinder-cDOT-NFS-IPv6/8175074/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/37/768837/1/upstream-check/cinder-cDOT-FCP/de48a9f/ : FAILURE in 3h 40m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"0e446961aca28ed7495df2b4b7195d2fea2a471d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-06 09:25:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/37/768837/1/check/DellEMC_SC_FC/a3861b2/DellEMC_SC_FC/181 : SUCCESS in 2h 18m 23s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/37/768837/1/check/DellEMC_SC_iSCSI/0473d24/DellEMC_SC_iSCSI/181 : SUCCESS in 1h 21m 29s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"492b4d8f1dd7c54b28963bcbdd2680d4b7ffab01","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-06 09:57:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/37/768837/1/check/DellEMC_PowerFlex/e08a349/DellEMC_PowerFlex/193 : SUCCESS in 2h 16m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"ed7fb65ce6294b343de7fcedadfa02c6aaea3296","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-06 10:36:59.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4365362761c9356b89b48c1977422ddb9825d73c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-06 15:37:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/37/768837/1/check/DellEMC_VNX_iSCSI/4a3247b/DellEMC_VNX_iSCSI/237 : SUCCESS in 3h 01m 09s\n- DellEMC_VNX_FC http://publiclogs.emc.com/37/768837/1/check/DellEMC_VNX_FC/cd40ccf/DellEMC_VNX_FC/238 : FAILURE in 1h 14m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"48bf5cd5aea2790b3f7beb98747651f58909ba8c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-06 23:30:58.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b5591183c0581e02fbdb4609a1b939be5f4c8577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-02-15 08:01:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c5f4f4214e2bc1d6dbcea6819a3c61361a6a7b7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-15 08:29:14.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/2/check/tempest-dsvm-full-focal-py3/79c92d5/ : FAILURE in 23m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/2/check/devstack-plugin-ceph-dsvm-focal-py3/c4a4ac4/ : FAILURE in 23m 21s","accounts_in_message":[],"_revision_number":2},{"id":"864c93123ab851c155254477a5c57529140444c0","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-15 08:57:38.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/37/768837/2/check/dsvm-tempest-TOYOU-ACS5000-driver/519ebca : FAILURE in 54m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f8dbf68d37a290619393b01b394b3cdeeab2746b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-15 09:12:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/37/768837/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a531819/a531819.html : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":2},{"id":"77916752393720b8501e2ff2f63ec28a36b03f0d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-15 09:19:04.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768837/2/iscsi/ : FAILURE in 1h 16m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/768837/2/smb/ : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":2},{"id":"47587e31dc3205acf34b3ed1b7771f9c51da5b5d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-15 09:20:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/37/768837/2/check/dsvm-tempest-zadara-driver/144f709 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f523a7541f90b33e40d3ec9bd460c761850f0353","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-15 09:23:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/37/768837/2/check/iscsi-driver-primera/0d05d70/0d05d70.html : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":2},{"id":"325cca628583a1bdaf0096757d557036d7257aa9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-15 09:48:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/2/check/dsvm-tempest-cisco-zonemanager-job/d1cad14/ : SUCCESS in 1h 18m 38s","accounts_in_message":[],"_revision_number":2},{"id":"ef633d1e0bfc45d4b6150aafc40dc2081224e0a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-15 10:06:04.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"3033b5cdbad3e622ed5b6607f92fa18f088863cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 10:37:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/676dd5cf632742bf9d80ac908613b143 : SUCCESS in 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6eab74111be4db29586aa953b657656 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72712ca53f9e4ca283f27e1ad580e183 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c5c5614063f494b842ecbd90f9771b5 : SUCCESS in 11m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9273cd1330f843fa9675af51cd364754 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab4d90453474271b5347fa5c4c85ad0 : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/51818bcfb1b2479b8dde477df2cd2c6c : SUCCESS in 46m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c99a2e9b7ec42589a7d23e087a62e5c : SUCCESS in 1h 52m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/508e554ce1b84a73a84bfa7f48673c20 : SUCCESS in 13m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8bcaa968923c44a29850eda4a2d915ce : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65d98724e0ed463b9949c951e5e66354 : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f4d51563a9e45efb0c2febdcb9e0a61 : SUCCESS in 7m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d2120b127aa4441b04125556f5b2953 : SUCCESS in 31m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5794270bc0540819f7f1c263bb8fe69 : SUCCESS in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f8182cd80ef9495782bfc3dc57610531 : TIMED_OUT in 2h 08m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6605d9eaf1b04318889849d7a7e1794c : FAILURE in 1h 21m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/634c276793e84b8ab2b2500b910df175 : FAILURE in 2h 32m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc666290456f43869e5580f0d9a26a49 : SUCCESS in 1h 07m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ed4846c95a85435ea302716c8829ca4c : SUCCESS in 1h 19m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f516784a5f2441e5bb1c319d09ab7261 : SUCCESS in 1h 19m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6592a8803af3412baf6d75b9ce1049e5 : SUCCESS in 1h 59m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b446c3b3203475496c2754b68732a46 : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a481936d726ea69712842f0a622d410c1b79037f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-15 11:04:41.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-37-768837-2/ns5_nfs : SUCCESS in 3h 2m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-37-768837-2/ns5_iscsi : SUCCESS in 2h 54m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"1725c0564260d05f46fdfe1263b398c76d62f03a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-15 11:06:25.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/37/768837/2/check/DellEMC_XtremIO_FC/7632324/DellEMC_XtremIO_FC/754 : SUCCESS in 2h 03m 15s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/37/768837/2/check/DellEMC_XtremIO_iSCSI/d64d289/DellEMC_XtremIO_iSCSI/756 : SUCCESS in 1h 38m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"7d38fe5e1b5b128a2872f51ee08a8afd92f2f0db","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-15 11:07:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/37/768837/2/check/DellEMC_Unity_FC/fd4dfba/DellEMC_Unity_FC/693 : SUCCESS in 1h 37m 35s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/37/768837/2/check/DellEMC_Unity_iSCSI/69408d4/DellEMC_Unity_iSCSI/655 : SUCCESS in 1h 59m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"3d213813e6de36992a79ef623252e4fd20f7e47e","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-15 11:43:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/37/768837/2/20210215182218/NEC-ISCSI : SUCCESS in 2h 13m 56s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/37/768837/2/20210215182218/NEC-FC : SUCCESS in 2h 20m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c1f6d733333872b31165586b8b02512d8d99d6d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-15 15:23:11.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"976c59687d3cea59b9ca0e49816133f0df69aad6","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 07:15:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/37/768837/2/check/DellEMC_PowerMAX_FC/9dc1488/DellEMC_PowerMAX_FC/468 : SUCCESS in 5h 20m 22s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/37/768837/2/check/DellEMC_PowerMAX_iSCSI/a1d95d5/DellEMC_PowerMAX_iSCSI/432 : FAILURE in 4h 37m 23s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"dfefd01db525435fce3978f3deed7bd80542d4d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-18 11:14:42.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ed2a9d0992c3a57bd8b7546afcd414e2e9ab69bd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-19 14:10:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\ntest job failures are unrelated (but annoying)\n\nnit about the commit message, otherwise, this looks fine","accounts_in_message":[],"_revision_number":2},{"id":"c68666f130d34da79d345ad7e0258ae657236063","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-19 14:10:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4d65e2088991925b3058d46dd118426aae124fb8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-19 14:11:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f67dd9eca7145efdeee22620c55f3f45034f7fb","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-19 14:15:05.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768837/2/iscsi/ : FAILURE in 2m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/768837/2/smb/ : FAILURE in 2m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9c60581278f26cf2f979f0678d8e59d2bfb8ab0d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-19 14:57:58.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/2/check-not-vote/ext-cinder-zuul/db3e6f7 : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":2},{"id":"32d79f929a1d2843cd336bdde2bc8a911a010af2","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-02-19 15:21:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cd512200e94d25afa205b56f5bef7c89d8ef637","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-19 15:33:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/2/check/dsvm-tempest-cisco-zonemanager-job/d62146c/ : SUCCESS in 1h 16m 56s","accounts_in_message":[],"_revision_number":2},{"id":"337f76984c53d58fdfd291d142b7eb61b7a34bc9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-19 16:30:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/83b9d93/ : SUCCESS in 2h 11m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/071bb79/ : SUCCESS in 2h 12m 22s","accounts_in_message":[],"_revision_number":2},{"id":"cbfef639efccf2e60714b5162f5bed7da2a528b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:02:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c55745237903474095626bd5e341d1f9 : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2378d0138f8947d69040ed6731af540a : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d915e9231c984a66b2e882c3244ef349 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97bdbcc8b5944eb6adb9a1d1eaff4971 : SUCCESS in 9m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2947c4656e714aecbe4a4e87ceea06df : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060779708c4e48d98f1ec5212fde0b28 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/5034417295af465aaeb66e5a5cbd6425 : SUCCESS in 58m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d525399f79d645828678f042ae47fcdc : SUCCESS in 1h 17m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3eb6577d68674a6ebd01ccc771b6bc7c : SUCCESS in 13m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4212b072a40f43339cc9b1d96de5e628 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d65c927f93f24abcafa9b0d3c0d24b29 : SUCCESS in 6m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/96e93064c750449caed590f3b6e792de : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/95e3bcafcf0d46dd9d430697f45831c8 : SUCCESS in 49m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fdfc5891e58f4b458222240ab8719f91 : SUCCESS in 8m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ab10155c1a1f40149c62d9ac25a6ca38 : SUCCESS in 1h 33m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6d24ff98c07a462a9dca9a27330b7f40 : FAILURE in 1h 36m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc6d43fca1a54908b702bbcbcaee4455 : SUCCESS in 1h 33m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/646d25056bb44e58815e063e09624160 : SUCCESS in 1h 14m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f84f16252704e3c973011eaad7852c6 : SUCCESS in 1h 23m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bf11df1e03884ceabd4350699097d4e5 : SUCCESS in 1h 38m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cd50d8b49f48469f99cfd843e6d0b48d : SUCCESS in 2h 20m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2aa78b7a63734664b3156adc900ae621 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":2},{"id":"19addbb288281bfd3d04dc6df9dbcf0793186ac8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-19 17:55:07.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/37/768837/2/check/DellEMC_PowerStore_iSCSI/53a5934/DellEMC_PowerStore_iSCSI/450 : SUCCESS in 52m 07s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/37/768837/2/check/DellEMC_PowerStore_FC/2ddb017/DellEMC_PowerStore_FC/444 : SUCCESS in 47m 06s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"70866cec605d12953e29a4ff6ce0782371ba2140","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-19 18:04:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/37/768837/2/upstream-check/cinder-cDOT-iSCSI/ce24671/ : SUCCESS in 39m 18s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/37/768837/2/upstream-check/cinder-cDOT-iSCSI-IPv6/0a9c96e/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/37/768837/2/upstream-check/cinder-cDOT-NFS/5eeee27/ : SUCCESS in 40m 15s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/37/768837/2/upstream-check/cinder-cDOT-NFS-IPv6/746d852/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/37/768837/2/upstream-check/cinder-cDOT-FCP/03227a9/ : SUCCESS in 1h 01m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"49bbfdbcb050ec566d36d0f3c7c05e1ef921f790","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-19 18:23:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/37/768837/2/upstream-solidfire-check/cinder-solidfire-iSCSI/df8adcc/ : SUCCESS in 1h 18m 49s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"59a9c5aae0174441cf0d93af75234f293148c63f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 18:23:26.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/37/768837/2/check/DellEMC_VNX_iSCSI/7aa93a8/DellEMC_VNX_iSCSI/420 : FAILURE in 1h 20m 30s\n- DellEMC_VNX_FC http://publiclogs.emc.com/37/768837/2/check/DellEMC_VNX_FC/76fabc9/DellEMC_VNX_FC/439 : FAILURE in 59m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"6eb25d09cfed03e2ca6b086246b52db678cbe22f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-19 19:09:28.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/2/check/tempest-dsvm-full-focal-py3/1da7910/ : SUCCESS in 1h 52m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/2/check/devstack-plugin-ceph-dsvm-focal-py3/61f8379/ : SUCCESS in 2h 19m 53s","accounts_in_message":[],"_revision_number":2},{"id":"0d64ed28ab27aea7d2466ebfc902c874b4087272","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 19:42:18.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"19eb6592fe0789a5d6f88895d29b55f2c5f871b2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 22:34:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/37/768837/2/check/DellEMC_PowerFlex/d0a66cf/DellEMC_PowerFlex/418 : FAILURE in 2h 49m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"fdb4fe6eaecb14c940e8761a0fee0235c05b2988","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-20 06:40:48.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/768837-2/PVME-iSCSI : SUCCESS in 50:41.68\n* PVME-FC http://os-logs.tristero.net/dell-ci/768837-2/PVME-FC : SUCCESS in 47:45.65","accounts_in_message":[],"_revision_number":2},{"id":"f001edf13f72f0b38ea6cbd49170c413d0579f62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-20 08:46:39.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"975a041f4af52a9235f290aad5f4a0e04f72c532","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-02-20 18:42:49.000000000","message":"Patch Set 2:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/768837-2/Seagate-iSCSI : SUCCESS in 45:19.40\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/768837-2/Seagate-FC : SUCCESS in 39:06.12","accounts_in_message":[],"_revision_number":2},{"id":"12567af5221756042159d415bbc55d39e86fd75f","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-20 22:31:38.000000000","message":"Patch Set 2:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/768837-2/Lenovo-iSCSI : SUCCESS in 47:24.63\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/768837-2/Lenovo-FC : SUCCESS in 44:33.14","accounts_in_message":[],"_revision_number":2},{"id":"809d4490842f3787d9c4062441b09241485da444","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-21 09:06:16.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/37/768837/2/check/DellEMC_XtremIO_FC/cd754fd/DellEMC_XtremIO_FC/821 : SUCCESS in 1h 57m 25s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/37/768837/2/check/DellEMC_XtremIO_iSCSI/5a62bb9/DellEMC_XtremIO_iSCSI/823 : SUCCESS in 1h 37m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"015abb4f5146cbf628a3ff46395e9425d74319da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-21 10:28:09.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a227ce9521e3a675f1d39682e72db4488048d744","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-21 11:41:30.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/768837-2/HPMSA-iSCSI : FAILURE in 1:05:13\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/768837-2/HPMSA-FC : SUCCESS in 39:53.94","accounts_in_message":[],"_revision_number":2},{"id":"419e45c45b3e46c1535e40062331f7d563a83d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-02-23 02:24:01.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"591ac4dc76f5a2e998f42bc578ae1fc169707608","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-23 02:24:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d46b054c38239b78f6a31b23b0556cf1627aaf4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-23 03:08:18.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/3/check-not-vote/ext-cinder-zuul/81d4567 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f5b08841c3a2ea777a8ebd40eaaeea7b013e93a9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-23 03:21:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/37/768837/3/check/iscsi-driver-primera/9e61947/9e61947.html : FAILURE in 55m 57s","accounts_in_message":[],"_revision_number":3},{"id":"70de3cba0c8d67f3ce658626cd9e42bb42325320","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-23 03:25:08.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768837/3/iscsi/ : FAILURE in 1h 00m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/768837/3/smb/ : FAILURE in 1h 00m 08s","accounts_in_message":[],"_revision_number":3},{"id":"87d498c85574f08a94cac3081da9490a9ecdaa9e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-23 03:32:15.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/37/768837/3/check/dsvm-tempest-TOYOU-ACS5000-driver/9090dcc : FAILURE in 1h 06m 16s","accounts_in_message":[],"_revision_number":3},{"id":"fcf38a254ef5428173d5112ac1e716c148fe4fb1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-23 03:35:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/37/768837/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e0fd748/e0fd748.html : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":3},{"id":"08e45fae279741ff27f28cf85f095856bd8af7ef","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-23 03:46:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/37/768837/3/ : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":3},{"id":"2567cb99aa231f96d94683e39f1901fd2b689651","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-23 03:50:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/3/check/dsvm-tempest-cisco-zonemanager-job/d921b16/ : SUCCESS in 1h 20m 36s","accounts_in_message":[],"_revision_number":3},{"id":"36355a30ae289df275f02784f1e8c65441718811","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-23 03:58:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d3bbf49/ : SUCCESS in 1h 32m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6aa953c/ : SUCCESS in 1h 32m 52s","accounts_in_message":[],"_revision_number":3},{"id":"203039712e89e5a5f5ae2691093e76c12e822981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 04:29:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26b0e8c195f7487fa6745a85c61e1fb0 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6ac8cdd10e3485ba2ff7da9d80fdb4f : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f53e0d833fa4b43ab027f5c79113b94 : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/246d2d1ee4ad4b76a89500c0aaf13d56 : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab02ee063ec14a709ad2f9e94bc19e13 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea9fa65c1aa43f3a17907e4beed84f2 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/65488584f7d3428e87babc2d6be0ed76 : SUCCESS in 58m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8bf29c8b7a04f57b54669cbfb493fb5 : SUCCESS in 1h 16m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a7018990766944788ced6a38fa71ceed : SUCCESS in 12m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2908a9903423457bac6c0065409f04f1 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/013ce5bbedee41178c289c2616b35d54 : SUCCESS in 6m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1070f342f175498495cdd452e238efde : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/de0a350e1429488e947f1ebcd31bc8b9 : SUCCESS in 38m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b60664d363946c6a210385369337b27 : SUCCESS in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d8fdc23c51249a3b689ed70fd4ef089 : SUCCESS in 1h 10m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1362430a60b04485b310b91c05d7aa06 : FAILURE in 2h 04m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/303848c62eeb4164bd0145b49d5e0a61 : FAILURE in 1h 31m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b8c5ef958b34353a81418b6a0929515 : SUCCESS in 1h 09m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9e9690c00e474d349b3748ffb0c3b8af : SUCCESS in 1h 45m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f866e534794d45208054ab863d93c0dd : SUCCESS in 1h 29m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c70068c57c504e7cb1d8208446135981 : SUCCESS in 1h 48m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a8a9a807fdc43489278991b14f2d268 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e1d374a805593f6f03dbdab7ac6a4bc1242032fc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-23 06:11:42.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/3/check/tempest-dsvm-full-focal-py3/497361d/ : SUCCESS in 2h 53m 58s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/3/check/devstack-plugin-ceph-dsvm-focal-py3/c6f4871/ : SUCCESS in 2h 02m 12s","accounts_in_message":[],"_revision_number":3},{"id":"dde2a811931cc955282f121b9d93d0a41ade3fd3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-23 06:53:01.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-37-768837-3/ns5_nfs : SUCCESS in 4h 12m 37s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-37-768837-3/ns5_iscsi : SUCCESS in 4h 1m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"b6538fb1cf2881c3bd60b64d26dde16452dfec21","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 07:43:54.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"953d4bcf308e5b6c18e4940aad12ea922cecb3dc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-23 11:04:52.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/37/768837/3/check/DellEMC_Unity_FC/73c4b2a/DellEMC_Unity_FC/759 : SUCCESS in 1h 39m 23s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/37/768837/3/check/DellEMC_Unity_iSCSI/c13b502/DellEMC_Unity_iSCSI/725 : ABORTED in 3h 01m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"965ba28adc17758f99e806ad85635ca9f367a8d2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-23 11:51:38.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/37/768837/3/check/DellEMC_XtremIO_FC/a1cf635/DellEMC_XtremIO_FC/843 : SUCCESS in 1h 58m 21s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/37/768837/3/check/DellEMC_XtremIO_iSCSI/36701f5/DellEMC_XtremIO_iSCSI/847 : SUCCESS in 1h 39m 48s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"59ee7cffcaed8a12eba7347e76dc6962e669624e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 14:18:55.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"be7a9f982d4f7f13c8bd71d7dce2f1f480dc1e59","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-25 03:35:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/37/768837/3/20210225101710/NEC-ISCSI : SUCCESS in 2h 15m 00s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/37/768837/3/20210225101710/NEC-FC : SUCCESS in 2h 18m 31s","accounts_in_message":[],"_revision_number":3},{"id":"0effb31db135f6ac2f1ba9a8e4bc6891e83369b7","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-02-25 11:44:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3df6d3f9b50c620cd3c53db58955004f06efb7e8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-25 11:45:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37536066e303ca2b920a31e08df5104b2036bc5b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-25 15:32:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/3/check/dsvm-tempest-cisco-zonemanager-job/58d4d0a/ : SUCCESS in 1h 18m 55s","accounts_in_message":[],"_revision_number":3},{"id":"8a7b8efc8eb8007205a6809594cf4e8b7e549e48","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-25 15:41:05.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/3/check-not-vote/ext-cinder-zuul/8337f5f : FAILURE in 7m 59s","accounts_in_message":[],"_revision_number":3},{"id":"99f2ecdf2693326d65718da66eb39b4350cfb1cf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-02-25 15:49:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9f53f76/ : SUCCESS in 1h 53m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0bb6d6d/ : SUCCESS in 1h 42m 45s","accounts_in_message":[],"_revision_number":3},{"id":"a374328da4a7140ed462d2e29fdb0b65a1266586","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 17:05:11.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"5024dc6d83470be4cc554eb022069b7f4065aebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 19:48:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45cb871affaf4868829c7b6deefa941b : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d97e8c4022e4ee88b13298b60484944 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/366ced5aebbc4b968f0d9b8bb228f243 : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76cc373296154a499e77f0342c301fa8 : SUCCESS in 13m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a909e2c78db48608b09e65bcc88cc8d : SUCCESS in 13m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e2628ab2d148c39eeef2c211de8e14 : SUCCESS in 11m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/d08cc01d0cfb4126b4197c85bdf180c2 : SUCCESS in 1h 01m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b389599e2bc46a6bf0816fdf86fbfcc : SUCCESS in 1h 11m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c99b7fe9a2ad4beeb7e3f74a992dd96d : SUCCESS in 15m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31311f1383d04a9db042710937213371 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9bdd625e94f649b987eefe0706e93512 : SUCCESS in 6m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f19e3daacba543d2a6e6262d045be03c : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9b08f2b8d6604090b02c9602e8a7eb80 : SUCCESS in 41m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ed1841414b24ba0875ff8f0f5c17a7f : SUCCESS in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4cb0b7ffe1544524ad779f14a3e79d1e : SUCCESS in 1h 15m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0a76bbd0da9e4e1285211cd5be407648 : FAILURE in 1h 36m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aa229a1e620d4e51a723a25addf38993 : FAILURE in 1h 31m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/945c0fa7600241a9948bca27e12bd370 : SUCCESS in 1h 11m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/549831efb0a549529de19da39be685d1 : SUCCESS in 1h 44m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/af0a70cedc9d4e19a2cca4399a7f7f47 : SUCCESS in 1h 41m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2ea49fb7e534aa5b8608d1efbf37303 : SUCCESS in 2h 21m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eabc8421257741039a00958e43c97d06 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":3},{"id":"11e0756a46feeb88483ec3dba10e0458a77e97fd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-26 02:22:57.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/d5650c1b4d7e4c9d92532aac47a5a93f : POST_FAILURE in 2h 02m 24s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/302f39be0ee64fc5ba5e4c7a4a5ccb5b : POST_FAILURE in 1h 58m 29s","accounts_in_message":[],"_revision_number":3},{"id":"8ab0bae4bc33f0a05fab1bea28b6f2b0f95091cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-26 12:26:04.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"4d6a72f4dc94d0d846147f3104fdd8fa5e8f0440","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-04 01:49:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8a839cb6286b1334f87ae7c6670fb6dc6bb6bc1f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-04 01:49:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e35063de9972829268dc4d93459fd7708874de02","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-04 02:29:25.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/3/check-not-vote/ext-cinder-zuul/9fb9cd0 : SUCCESS in 39m 28s","accounts_in_message":[],"_revision_number":3},{"id":"c9b67e538774b7798b2948d4a9ac76007694750c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 04:01:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99a4144631ee466ca95e102aac5c814a : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b2f65e1cfcd41bf83171f25e0886a1a : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/908bdceabd6d460b80f12cdaf51d37ed : SUCCESS in 11m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7de2055509154664b527a309f83b755c : SUCCESS in 12m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcc85a5e510643bd9e6292dc09b93406 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a92a106ea6a34abc8d8fe54d4975ad1d : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/55c62c59ea8b44f3b032a73fe3ae7f61 : SUCCESS in 1h 06m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c71768272664562a99723d5ee9eef5f : FAILURE in 1h 09m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6040a823ae9d4503a1b49c8d0290724f : SUCCESS in 13m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ef23de1b78d4d528a7b6905b59e8cde : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d47786b8b9c4dfeb41b5d0fb82dc54b : SUCCESS in 6m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c02b874b0ca492b81b6de9136a41538 : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/945fadcbbe384d18b5ce471403cf8111 : SUCCESS in 50m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/852d509ae3484cfd9f52d90dfb4ffb5a : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3af897c814014cb7b2ffa287340c8b2e : SUCCESS in 1h 35m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c13a5a9c6b134c9c9545ae104856bec4 : FAILURE in 1h 25m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/85fb7c96b1a6421e8361a9be0d909f75 : SUCCESS in 1h 58m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d63d194fd3424db0b0a95a4e0bf7322b : SUCCESS in 1h 14m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/042fe1bb298c4d3a8a57788b8530d13e : FAILURE in 1h 19m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/460c025ee2c04d0a91652563ffe9b974 : SUCCESS in 1h 25m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54bd62d00efc43d1b9626c59d8f61cb2 : SUCCESS in 2h 09m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d7807826d5d4a63878a1446e25a0194 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":3},{"id":"099fc61e6a6673bbdf61fb617c0d764c5cdd98ce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-04 06:25:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/3/check/dsvm-tempest-cisco-zonemanager-job/086f888/ : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":3},{"id":"b59a8f16b854a76c847f2676eb6ccaaa8dce3b25","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-04 06:38:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/665760c/ : SUCCESS in 2h 26m 16s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f24d0eb/ : POST_FAILURE in 1h 55m 58s","accounts_in_message":[],"_revision_number":3},{"id":"050b4e577d700f463a7831906c5346c66c6b5755","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-04 12:59:34.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"623e9ede7a82707829432d632a7950db73797ed4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-04 18:07:45.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/0df40dee280d4146985bae50e4a00461 : POST_FAILURE in 2h 01m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/dca1eb28f60c41a38a6aa0948df31d2f : POST_FAILURE in 2h 06m 16s","accounts_in_message":[],"_revision_number":3},{"id":"2f75a8838d67d67da73d124815d4e4a261220c89","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 02:30:55.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"e2b55cb480dbc246c1ba75c1ad21bebec39d7af7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-07 16:48:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4ec24d63996cd24ba0dfb5a152070819d3ebde50","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-07 17:28:05.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/3/check-not-vote/ext-cinder-zuul/5b1d656 : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":3},{"id":"74a8c930b9224f009a280cf12926b390521a6160","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-07 18:13:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/3/check/dsvm-tempest-cisco-zonemanager-job/8dd2e1f/ : SUCCESS in 1h 19m 42s","accounts_in_message":[],"_revision_number":3},{"id":"37ebd87f69ac3789702df6f1309dee99e00c6711","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-07 18:38:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/37/768837/3/check/dsvm-tempest-zadara-driver/8813b39 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 48m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c2d17f6b61bf857476e9d43cf303dc67e5bd8b98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 18:55:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b74e5332c4554a48a51361888b1438bf : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c1d602bea841ebacb133e2fbc94325 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c2d7e1d10cb46afb59f95d3dcbefe00 : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99a845d50a594b4cae53aca83cbcabf9 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/165bf1730ea8445a896a985a08ec0bfc : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48fa23b30357474892f7bc354e357f47 : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/0eebf95854d9464b8cbdc48e4ab2a8e0 : SUCCESS in 1h 02m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/281269ef61014e12af871e2ecd0a3182 : SUCCESS in 1h 14m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/be0a9f772c5843a0a681c6628764d5df : SUCCESS in 12m 48s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fdbec69c01734c0897ee7ee03c5dceb2 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9fc0d4888464f5f9349ee90c6eab23e : SUCCESS in 6m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a1cf9a1820364ed0b960424602a8affd : SUCCESS in 10m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ff62fde419894233a233e9ede5c2cba2 : SUCCESS in 38m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c2c203a214548f697d792d01508fe3f : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f579f528f4f0431b9bfb2003ac55d286 : SUCCESS in 1h 46m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1c1d280833ef414abeba1ee2400f8a34 : FAILURE in 1h 24m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de7a04a6786e47bc81d27bcb0372439a : FAILURE in 1h 30m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5de7e0a9796843ddbed1f6b4fc39e1b0 : SUCCESS in 1h 03m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d187520b874545579617893abe725d06 : SUCCESS in 1h 15m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a42d2dc79f16482eb7c9418cd44c5e05 : SUCCESS in 1h 15m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56dabcaf94564f95a70f8153f2606f60 : SUCCESS in 2h 05m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c136f8718b1b4502be93d23bfac62e4a : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":3},{"id":"81629f816db098cd4adb8e06bebcf2960e7b4a1c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-07 19:05:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/3/check/tempest-dsvm-full-focal-py3/b50fea3/ : SUCCESS in 2h 03m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/3/check/devstack-plugin-ceph-dsvm-focal-py3/5bbe8e1/ : SUCCESS in 1h 59m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f46d4de283fce5460f64643e80eb34591df82ab4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-07 19:11:21.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2f71dc5a9a0b81d8962af6a8112ec7f32ce9f711","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-07 19:53:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d103da3/ : TIMED_OUT in 3h 03m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f997e37/ : SUCCESS in 1h 33m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a401642b5a2e2bcad4a8e0705ab9481b1eddaaf0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-03-08 03:19:23.000000000","message":"Topic set to bp/six-removal","accounts_in_message":[],"_revision_number":3},{"id":"7d9c70e995dcfb852baafd358683e5b993b833fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-19 14:46:35.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1d78182af00ee8bb5f69a2e889213d2f26079a27","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-19 14:52:49.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/37/768837/4/check/dsvm-tempest-TOYOU-ACS5000-driver/9ccb367 : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":4},{"id":"5a132a0e3ada45fdd56967d33ee62ba7acaf09fb","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-19 15:20:38.000000000","message":"Patch Set 4:\n\nIf this passes the check it should be good.  We are trying to get rid of Six.","accounts_in_message":[],"_revision_number":4},{"id":"62a4ca2c156bb7c318ca827dcb6e483c14f4c5a5","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-19 15:20:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"39e25951b7c54ec53a97e0765d26218f58c14172","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-19 15:22:06.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/4/check-not-vote/ext-cinder-zuul/5012481 : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":4},{"id":"d40ecec1373aa2f32e2f536751a9f885fabf74f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 17:24:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/947cac3e8a7247a4b4337e1a61bc0803 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194b6d5b12194164b15cadbb2af82a29 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43bec4ac9d544f16880082f5e26828e1 : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec10bf1acd1b412ebd259591fd46b597 : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/671799e38b4342bb966f4b02111654c1 : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f5d5c56da75448cab6a975e66668491 : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/dd94cb239082451386d20393e2d3edb2 : POST_FAILURE in 1h 04m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4f5c298019644195b428e17a857b5e0d : SUCCESS in 1h 12m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/da1b531a1d504e36acd219e0c6aa357a : SUCCESS in 13m 07s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34829204b6a64261822feb39f2f55abd : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dfdc298cf51b40568b011efe7b48379c : SUCCESS in 6m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65815e8fc31c4f4bb22745433a7a79e1 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e9fc37a31b8e499a87a6a713acdd1332 : SUCCESS in 43m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/53f90b69677e4d56b96361cca99c12bb : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d170bd661af243868fc4fc5331b13cc0 : SUCCESS in 1h 19m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/380cb657cbd04743b7f8c7337eca69e8 : FAILURE in 1h 48m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0ad51453e92c491493b957ec59b32334 : SUCCESS in 2h 09m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3026549b46c1406f8ab9901a2d443aba : SUCCESS in 1h 36m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e342f7ff5ac74ef5b8d349c4acce3de4 : SUCCESS in 1h 35m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57f9343943bc40acbcedd488e4267c23 : SUCCESS in 1h 57m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70a6c39c06e94467ab763641689952f1 : SUCCESS in 2h 01m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/366bbdf1b587402eac918c1fa2e6f93a : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":4},{"id":"72a0d630ca3a2814767017fe1a14cc74fb06355f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-19 18:33:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a5dadb4/ : SUCCESS in 2h 13m 02s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3804dc3/ : SUCCESS in 2h 09m 59s","accounts_in_message":[],"_revision_number":4},{"id":"2d0ccae28637dd463b2c359f8cb8889b1a6e8bb4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-19 18:59:46.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/4/check/tempest-dsvm-full-focal-py3/2dea642/ : SUCCESS in 2h 10m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/4/check/devstack-plugin-ceph-dsvm-focal-py3/5093f20/ : SUCCESS in 2h 12m 15s","accounts_in_message":[],"_revision_number":4},{"id":"4f41c4804d4035186b07de7a24b763b2f91ad300","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-03-19 20:29:04.000000000","message":"Patch Set 4: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"01a98d5e69c9a2912d24d7e15cc53a17e51ed8bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 22:26:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4860e70668da40f68e6f5229497e67d4 : SUCCESS in 15m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e789f94b71fd4e43922bf1df7ee53ded : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56e88c8ca9fd477199ec21efb6811981 : FAILURE in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/197add50640a48d4b1f81c631f7ce9c4 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af800da1e6ec40b7a99a8268ee894187 : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2261a8003c83467792037102fa58fb5f : SUCCESS in 16m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/81090058ee9f45fa82c9a001c26d6be9 : SUCCESS in 1h 11m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47837b0ff5274a1196407025a10d097b : SUCCESS in 1h 10m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3de7b745ca6b49588d49e4e2e08ad372 : SUCCESS in 14m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe0e6fb09e5b41f5b2b8f5bf54a86a08 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19925e2b7f5b4ce4b07b632a2bdaed2f : SUCCESS in 7m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5cf02e8726fd4222b936ad7456df585c : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ccefbfa80364492b97bea40408ff8f36 : SUCCESS in 42m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fc59fed6ff0241f98f2e31a825e5a560 : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ad2e57ee16e4320b0aa1aa398859fcb : SUCCESS in 1h 24m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/17018dd088824e54bc80a70f17c5cd6f : FAILURE in 1h 20m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d7a2ca137fa4cd0982a800cc99c6174 : SUCCESS in 1h 30m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c1d1f623466a4fa583ea849daaf83885 : SUCCESS in 1h 11m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91d5997d77bd4d58936b26663604f49d : SUCCESS in 1h 27m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/90a5fcf66b40406eabd5556b44b682c3 : SUCCESS in 1h 34m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a80fecf4c9da4a21b560688bcf51c964 : SUCCESS in 1h 56m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1be6d41124e4973b41b34b7169593b9 : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":4},{"id":"b9b1930fcb38e211f90b35120d5776bb59d0d0ba","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-19 22:30:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/37/768837/4/check/dsvm-tempest-zadara-driver/a0f1ce7 : SUCCESS in 1h 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cfbc09458d117a91adc754b31d22d3e7489be397","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-19 23:17:21.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/4/check-not-vote/ext-cinder-zuul/6e3eb4d : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":4},{"id":"817c396fb4bc947114229f1c7740f1554453a26c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-20 00:36:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/4/check/dsvm-tempest-cisco-zonemanager-job/6940908/ : FAILURE in 1h 22m 20s","accounts_in_message":[],"_revision_number":4},{"id":"25909ab6f004262fc50bb404306463d87650d9b1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-20 00:55:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/69df77a/ : SUCCESS in 2h 25m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7755155/ : SUCCESS in 2h 09m 14s","accounts_in_message":[],"_revision_number":4},{"id":"51208047e6d68f2c36ca938b81ab73978c7dd749","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-20 01:24:49.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"1a72fcd51bef52c2431809f0750baa77fb4b9990","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-20 06:14:16.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/4/check/tempest-dsvm-full-focal-py3/3a31295/ : SUCCESS in 2h 17m 49s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/81bfbcf9ac75484b85d3294e146bc016 : POST_FAILURE in 2h 09m 24s","accounts_in_message":[],"_revision_number":4},{"id":"a84a23315ca71b432685820746d95d53b6c9d9ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-20 12:35:18.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"4deee221df894e622b2b765e62eebd8507a03634","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-03-21 04:25:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/37/768837/4/check/iscsi-driver-nimble/cac02ea/cac02ea.html : FAILURE in 1h 25m 51s","accounts_in_message":[],"_revision_number":4},{"id":"25e0873b31fc6a47a0ba50c8d71460b5060dc2df","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-21 15:12:21.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-37-768837-4/ns5_nfs : SUCCESS in 4h 46m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-37-768837-4/ns5_iscsi : SUCCESS in 4h 29m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6adb9c6c910bc6d51ff99827ddde14230383a979","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-21 18:43:53.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"61c4445f37fe1c69e499c1d25847c3b918deb4dc","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-03-21 20:30:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a6ea1414722c1c3b14831a9ed43fe5484142a3cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 20:49:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8be33c579aa24876a98b73677f72312b : SUCCESS in 14m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ba97cd7bcbe40f2b192bbf2965f7e80 : SUCCESS in 15m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fa0cf8f6ce7bdb01948e4b98610225614baee5a6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-21 21:10:27.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768837/768837/4/check-not-vote/ext-cinder-zuul/031159d : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":4},{"id":"fab30e507abfa10b17f33f7e5f96847c7c948756","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-21 21:43:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/37/768837/4/check/dsvm-tempest-zadara-driver/86f4fe2 : SUCCESS in 1h 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b3af0d546613efa895c2a75a726e555cb647ac77","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-21 21:57:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/4/check/dsvm-tempest-cisco-zonemanager-job/5505bfc/ : FAILURE in 1h 21m 36s","accounts_in_message":[],"_revision_number":4},{"id":"acfcd8238838c16d6fea7caa37f1334774f0863e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-03-21 22:25:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b19144d/ : SUCCESS in 1h 53m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8faef57/ : SUCCESS in 1h 35m 18s","accounts_in_message":[],"_revision_number":4},{"id":"65dcd111919e7013942694917b2ed96593d46a6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 22:28:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62969adfcfb54d04a9060cc2f70fedae : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73471dc5db25492f82382af392a90576 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18de8c4a5bc94a0f8f0df05dad63d1f8 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfa130a04b674147842363a33a124c68 : SUCCESS in 9m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0592be3d153b4e3882215363cd4da29f : SUCCESS in 15m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba1655e58bc4dfaa0261c785fd76cbf : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/cbba2b6240e34c7997f9e5d335f0b6b7 : SUCCESS in 1h 02m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8829640c5794ee4b9285be602993064 : SUCCESS in 1h 07m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4395c3a45051428eaf251efb630ded09 : SUCCESS in 12m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2975a39becdc4eb38d0f5d622036e6da : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94159141e9574b1f9adfdeaff12fd52d : SUCCESS in 6m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/481a491f96764394b1cdc614d5bc1582 : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c726317a0fe94b878a7fb840d57fa5b0 : SUCCESS in 34m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a49e1108e0c4218aed7143de20d4139 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51cac17dc640469db3db8fa5d9f8c35d : SUCCESS in 1h 20m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/708ea983898e49aa82b52bb1e23d49e6 : FAILURE in 1h 43m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae3e7f5d75a04293b437c0dbe63706aa : SUCCESS in 1h 29m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/16a7ea40f9654da08dbc8e01e11351a1 : SUCCESS in 1h 08m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eef60d07ec0a46a692d740724fc25115 : SUCCESS in 1h 21m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4afc1c873c574e0f89948b6de6c7cdcc : SUCCESS in 1h 23m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8199bfca25e34b3d9f7d0632eb2c5b1b : SUCCESS in 1h 56m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c97499f93db4af9a3a8591b4b2deb8e : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":4},{"id":"8a8774ebdc93c82bf3e13f9659d88cc0907acb79","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-21 23:00:55.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/4/check/tempest-dsvm-full-focal-py3/75564c4/ : SUCCESS in 1h 40m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/4/check/devstack-plugin-ceph-dsvm-focal-py3/4a31c04/ : SUCCESS in 2h 26m 08s","accounts_in_message":[],"_revision_number":4},{"id":"097faaf6eb0c89b9a99b571ddfc4ee7f278569a1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-21 23:13:31.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/37/768837/4/check/DellEMC_PowerStore_iSCSI/b49b7a5/DellEMC_PowerStore_iSCSI/667 : FAILURE in 44m 14s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/37/768837/4/check/DellEMC_PowerStore_FC/52a4e1a/DellEMC_PowerStore_FC/661 : FAILURE in 38m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"631c272c93b0993c2b0ee27626c1a319253e89bd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-21 23:29:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/37/768837/4/upstream-solidfire-check/cinder-solidfire-iSCSI/6ccbd30/ : FAILURE in 59m 48s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"6db401f02d3cf5f0963d4b91f23f5c45ff6b4ce6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-22 00:43:53.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/37/768837/4/check/DellEMC_SC_FC/1951a6e/DellEMC_SC_FC/480 : FAILURE in 2h 10m 00s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/37/768837/4/check/DellEMC_SC_iSCSI/9b52e3a/DellEMC_SC_iSCSI/477 : SUCCESS in 2h 14m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"fa44c2cd915a36b88372cacbc1fa24def2c9f786","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 01:27:29.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/37/768837/4/check/DellEMC_VNX_iSCSI/e86b7de/DellEMC_VNX_iSCSI/611 : SUCCESS in 2h 58m 16s\n- DellEMC_VNX_FC http://publiclogs.emc.com/37/768837/4/check/DellEMC_VNX_FC/5b34085/DellEMC_VNX_FC/650 : SUCCESS in 1h 29m 24s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"067533fffb7d3a415b72a1bb941ddc62fdccae51","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-22 01:42:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/37/768837/4/upstream-check/cinder-cDOT-iSCSI/856153e/ : SUCCESS in 38m 50s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/37/768837/4/upstream-check/cinder-cDOT-NFS/f25eed9/ : SUCCESS in 41m 34s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/37/768837/4/upstream-check/cinder-cDOT-FCP/514b0af/ : FAILURE in 3h 13m 06s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"b7ee87bcfabf9f26ebe0f4719f25abfbb13791eb","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-03-22 03:25:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c6242557a7551fedc07ac505a7a594a0ff2adf82","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 06:52:00.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/37/768837/4/check/DellEMC_PowerFlex/7d2f3ff/DellEMC_PowerFlex/627 : SUCCESS in 2h 41m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"3e26a2920e893e737c02a69e2f9346b2e8e9d031","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 08:42:21.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"00b25e047d964879b7467c0fb6cf782be36c5d28","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-22 11:14:00.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/768837-4/PVME-iSCSI : SUCCESS in 1:03:37\n* PVME-FC http://os-logs.tristero.net/dell-ci/768837-4/PVME-FC : SUCCESS in 56:53.01","accounts_in_message":[],"_revision_number":4},{"id":"c04dbe5a24757bb528d68d63b1c0e14a7fbe2407","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-22 12:19:57.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/37/768837/4/check/DellEMC_Unity_FC/3bd4a6f/DellEMC_Unity_FC/930 : SUCCESS in 1h 24m 04s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/37/768837/4/check/DellEMC_Unity_iSCSI/425d441/DellEMC_Unity_iSCSI/882 : SUCCESS in 2h 01m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"64d311718930e638f9315f1eb1459582e5b539b8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 13:37:55.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"5ec77c713e62772ea71dafbaadc685e295d178cb","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-22 21:42:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/37/768837/4/check/DellEMC_XtremIO_FC/193cf8c/DellEMC_XtremIO_FC/1095 : FAILURE in 2h 01m 45s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/37/768837/4/check/DellEMC_XtremIO_iSCSI/453aacb/DellEMC_XtremIO_iSCSI/1099 : SUCCESS in 1h 46m 48s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"18a01bdecf82f235821b2496a6e410919e8617a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-23 01:49:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"92c203310ee0b2306a60ad51dd1562452e390545","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-23 08:37:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/37/768837/4/20210323151812/NEC-FC : SUCCESS in 2h 18m 24s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/37/768837/4/20210323151812/NEC-ISCSI : SUCCESS in 2h 19m 24s","accounts_in_message":[],"_revision_number":4},{"id":"c2c4b901abbf3b86efc4fee9a97c1150bd4ce701","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-24 10:26:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5898f9e9d61cbff0902f89649f5f71cde2389183","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-19 16:56:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f7762f03d69132bd59a507acf33223da09143691","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 16:57:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8ff193005346d4edab383362df33be881fb7b284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 23:27:24.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10a6195b843f4fbba782616b84d42777 : POST_FAILURE in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d189aff5790d4a089fbf4f3ff8391bad : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84f4a365cd23454a945f5d6f0dcfd836 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bb18c26f02245318e745af90f766f06 : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39d3e0f347441f0a99f596924413349 : POST_FAILURE in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/cb55586208a6486fa48c1f9930adc63d : SUCCESS in 1h 01m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c4b53684f8645b29793da5dfeae5b70 : SUCCESS in 1h 17m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f9ba53cd18c14628a1e96d07e0a8c868 : SUCCESS in 1h 29m 49s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/039fb0116c9b4162bc8fb526e9eba771 : SUCCESS in 1h 23m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23e8c1ab6fdc4dd6a838a56caff22ccf : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":4},{"id":"bbec0c8bebda1f27fda2a09807d0e3970317d53e","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-04-20 03:27:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fcdbdf0ca527ebf86455745893b3bb9d1e1d0363","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 03:53:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0979bc491b8e4aba9e6bf0eebe1ee0e5 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/459f75367c224114a0dc37eca47fc70b : SUCCESS in 18m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62df59f157fd4eda5fdd702dfa64ced8be09a528","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-04-20 04:22:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768837/4/iscsi/ : SUCCESS in 53m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/768837/4/smb/ : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":4},{"id":"2b9e9f37a2515ea05f3d192f017b0816c3dde1d4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-04-20 04:45:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/37/768837/4/check/dsvm-tempest-cisco-zonemanager-job/13a862a/ : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":4},{"id":"142d70c317c021c5005e261df5b133817842533c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-04-20 05:11:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/37/768837/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3ec3761/ : SUCCESS in 1h 42m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/37/768837/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1ac8d17/ : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":4},{"id":"d20284ea4c15e0718e56eae97a0a07faf5aab939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 05:25:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f75d1eae84454107b932d2f4163a06cf : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1002413cfac485c8338ab364ff46334 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5ef5eae26d149248d9fe2c26ee52aca : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cda4bba07874929b3bbcb90951824f8 : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f3971c140d24ee988794986c13c2847 : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbabb9d747d4bd5832d06251089f73e : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/c1402afc1ed741a294c436a189baa1b0 : SUCCESS in 1h 00m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/201af92c00e84e33b5bae52e3b51f241 : SUCCESS in 1h 15m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/13931c21e76e4ae396f64bd24f29794b : SUCCESS in 12m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2e824dbc4421462285998785041f52c9 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92db6135b3554430afcb110d4b3a2f2a : SUCCESS in 6m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/87de9267b6df49b9a776aab1ddbac8ae : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b94371be88f549e19cede147bdcddeab : SUCCESS in 33m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/157ec0aec3fe4fb0a3abe184036740a4 : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/76e7661e2c1743f5a0aab7e31c36a281 : SUCCESS in 1h 03m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3a9820c284184658b8558ffd09517c2c : FAILURE in 1h 18m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bec70451e73c496f9cf1fffcde8e1f4b : SUCCESS in 1h 02m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c17b55042a304c4585b2b11ec32d86f6 : SUCCESS in 1h 06m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3c1c29a20b314e608179e7f72c15c372 : SUCCESS in 1h 30m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8ddfd655abe84eba9118e2880bafb38e : SUCCESS in 1h 38m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9791464ca2604e83a853d5d0c3e31532 : SUCCESS in 1h 52m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22765f5de3b04e429d818201edfb4e0d : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":4},{"id":"bfd38e00232af992a715ae77192d34ee773c83e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 05:25:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1a4e4486e55be165ec39ff46daf35e8afcd43bba","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-04-20 05:56:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-37-768837-4/1618896306 : SUCCESS in 31m 40s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"77311f978f9e7c1cc918a4a581b577166599c441","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-20 06:19:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/37/768837/4/upstream-solidfire-check/cinder-solidfire-iSCSI/e2e461f/ : FAILURE in 52m 50s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"ae9d2b2f7ae00c85554b80346270f47d9b9e77c2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 06:23:54.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"48feb3704d765e9c58d47a849d64a7922ee27a63","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-04-20 07:21:48.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/768837-4/Seagate-iSCSI : SUCCESS in 43:31.79\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/768837-4/Seagate-FC : SUCCESS in 40:12.30","accounts_in_message":[],"_revision_number":4},{"id":"f4b4a83af8f4c8a6d7fc88ffa136d822505aef1a","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-04-20 07:33:17.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/768837-4/Lenovo-iSCSI : SUCCESS in 49:19.94\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/768837-4/Lenovo-FC : SUCCESS in 46:01.25","accounts_in_message":[],"_revision_number":4},{"id":"4d11a957b711734ac7790f307f759a13394412b8","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-04-20 07:40:02.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/768837-4/HPMSA-iSCSI : FAILURE in 1:09:36\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/768837-4/HPMSA-FC : SUCCESS in 33:00.19","accounts_in_message":[],"_revision_number":4},{"id":"1d1f690503d08a81c5b5397db9b7676c5def4e79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 07:49:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1660819e573457fa44aecb4bbb6f476 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18970153c102481688621d2e1bf7ea62 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69374c48396b4e24af837358a913e41d : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e1661b1f4f4ba48c2edc8954e35aaa : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee9e4e5626444105a591c24ecfa255ba : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/5a8e55acddc049ccb114df4f4512f40f : SUCCESS in 1h 03m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c32d6de40904d6d906fc06c3a3fd5b7 : SUCCESS in 1h 03m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/743836cd91a740889a87410a96b2039e : SUCCESS in 1h 14m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2155588992034278874d0d37503e3761 : SUCCESS in 1h 19m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/843eac76c08d44e4865849668d3137a3 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":4},{"id":"ca9721173dd337414d4439098fc08bcb9a939b44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 07:49:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f7b4f4a10379ee43e6969713da7d5f59e4dbd316","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 07:51:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a037f90914f4587955958afbfa45ba8 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":4},{"id":"8f6f8d0f744ab67cfa5a438d80b1e1b464f0ecc7","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-04-20 07:52:39.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/768837-4/PVME-iSCSI : FAILURE in 1:00:36\n* PVME-FC http://os-logs.tristero.net/dell-ci/768837-4/PVME-FC : SUCCESS in 52:59.81","accounts_in_message":[],"_revision_number":4},{"id":"5aae24f4ffcc414d2a59f5531322c8584caf4567","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-20 08:50:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/37/768837/4/upstream-check/cinder-cDOT-iSCSI/c7bfa89/ : FAILURE in 30m 19s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/37/768837/4/upstream-check/cinder-cDOT-NFS/a9998fb/ : SUCCESS in 1h 32m 52s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/37/768837/4/upstream-check/cinder-cDOT-FCP/3fabe5c/ : FAILURE in 1h 06m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"74f86016c8285cc6f8ba5d5217cc4e6300543e17","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-04-20 12:04:56.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/37/768837/4/check/DellEMC_SC_FC/40275ad/DellEMC_SC_FC/6 : FAILURE in 2h 13m 44s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/37/768837/4/check/DellEMC_SC_iSCSI/7ca70f5/DellEMC_SC_iSCSI/6 : SUCCESS in 2h 03m 04s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"ef1565b9935355e8622b32ada4baacbe929319f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 13:04:44.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"001882e6a9b44da461c8a0c2e9da4afd8dde81f2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 13:36:34.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/37/768837/4/check/DellEMC_PowerFlex/ae3f93c/DellEMC_PowerFlex/6 : SUCCESS in 2h 31m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"cbcfdd321a72b3363530b727f5504792541da5e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 13:52:47.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"0acac9271698ac6061dfcbe5663ebacbaac30e02","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 18:18:48.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/37/768837/4/check/DellEMC_VNX_iSCSI/51e4350/DellEMC_VNX_iSCSI/6 : SUCCESS in 3h 30m 24s\n- DellEMC_VNX_FC http://publiclogs.emc.com/37/768837/4/check/DellEMC_VNX_FC/3dcc55b/DellEMC_VNX_FC/7 : SUCCESS in 1h 29m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"ae3a1f37290139a58666e25fecf3a3a96b796122","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 21:08:45.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"a04f989eee767fa1e9dac50b6d72c37e967a6d78","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-21 00:06:25.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/4/check/tempest-dsvm-full-focal-py3/2b5346c/ : SUCCESS in 1h 34m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/37/768837/4/check/devstack-plugin-ceph-dsvm-focal-py3/a7fb33f/ : TIMED_OUT in 3h 07m 44s","accounts_in_message":[],"_revision_number":4},{"id":"e6f291613999b7f54b7151acd65860bec9509dc3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-04-21 01:03:05.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/37/768837/4/check/DellEMC_XtremIO_FC/44a93ff/DellEMC_XtremIO_FC/20 : FAILURE in 2h 06m 36s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/37/768837/4/check/DellEMC_XtremIO_iSCSI/aadfe86/DellEMC_XtremIO_iSCSI/20 : SUCCESS in 1h 53m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"445285eccfb47c66dbfc79af3e56ad4ac6f48929","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-21 04:20:56.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/768837/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"83ff2ae9e8827cdf646f75111697089ad2ebd8fb","revisions":{"8fa4210a158bf79210e30deee300245db2c10d4e":{"kind":"REWORK","_number":1,"created":"2021-01-02 02:11:58.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/37/768837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/768837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/768837/1"}}},"commit":{"parents":[{"commit":"8240195bd6433cc8d718db76878ea3d244580bc1","subject":"Merge \"RBD: Retry delete if VolumeIsBusy in _copy_image_to_volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8240195bd6433cc8d718db76878ea3d244580bc1"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2021-01-02 02:09:25.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2021-01-02 02:09:48.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nremove six in quota.py and exception.py\n- six.string_type\n- six.integer_types\n\nChange-Id: I86bc7f14d7d9006e93b33f9453b5727c9bb26c8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fa4210a158bf79210e30deee300245db2c10d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fa4210a158bf79210e30deee300245db2c10d4e"}]},"branch":"refs/heads/master"},"0aa3ae0d89dfa6cc88354239e6b730ee5d059b86":{"kind":"REWORK","_number":2,"created":"2021-02-15 08:01:46.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/37/768837/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/768837/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/768837/2"}}},"commit":{"parents":[{"commit":"4bea91f1da54e837d371806230eb17f5b9c9f119","subject":"Merge \"Move brick calls from cinder.utils to volume_utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bea91f1da54e837d371806230eb17f5b9c9f119"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2021-01-02 02:09:25.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-02-15 08:01:30.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nremove six in quota.py and exception.py\n- six.string_type\n- six.integer_types\n\nChange-Id: I86bc7f14d7d9006e93b33f9453b5727c9bb26c8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0aa3ae0d89dfa6cc88354239e6b730ee5d059b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0aa3ae0d89dfa6cc88354239e6b730ee5d059b86"}]},"branch":"refs/heads/master"},"e3e5dceb5e35c78cd686226feef6704800e97b4c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-23 02:24:01.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/37/768837/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/768837/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/768837/3"}}},"commit":{"parents":[{"commit":"4bea91f1da54e837d371806230eb17f5b9c9f119","subject":"Merge \"Move brick calls from cinder.utils to volume_utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bea91f1da54e837d371806230eb17f5b9c9f119"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2021-01-02 02:09:25.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-02-23 02:23:28.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nremove six in quota.py\n- six.string_type\n- six.integer_types\n\nChange-Id: I86bc7f14d7d9006e93b33f9453b5727c9bb26c8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3e5dceb5e35c78cd686226feef6704800e97b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3e5dceb5e35c78cd686226feef6704800e97b4c"}]},"branch":"refs/heads/master"},"83ff2ae9e8827cdf646f75111697089ad2ebd8fb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-19 14:46:35.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/37/768837/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/768837/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/768837/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/768837/4"}}},"commit":{"parents":[{"commit":"4bea91f1da54e837d371806230eb17f5b9c9f119","subject":"Merge \"Move brick calls from cinder.utils to volume_utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bea91f1da54e837d371806230eb17f5b9c9f119"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2021-01-02 02:09:25.000000000","tz":480},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-19 14:46:23.000000000","tz":0},"subject":"Remove six from quota.py","message":"Remove six from quota.py\n\nremove six in quota.py\n- six.string_type\n- six.integer_types\n\nChange-Id: I86bc7f14d7d9006e93b33f9453b5727c9bb26c8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83ff2ae9e8827cdf646f75111697089ad2ebd8fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83ff2ae9e8827cdf646f75111697089ad2ebd8fb"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
