)]}'
{"id":"openstack%2Fcinder~768858","triplet_id":"openstack%2Fcinder~master~If395e63bd2cb22b5bc3f807755631b81ad3ac015","project":"openstack/cinder","branch":"master","topic":"bug/1909630","hashtags":[],"change_id":"If395e63bd2cb22b5bc3f807755631b81ad3ac015","subject":"Remove invalid quota resource when update volume type name","status":"ABANDONED","created":"2021-01-02 06:27:35.000000000","updated":"2021-08-16 07:27:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7792ea0753ded9a29d01cf654d2eda0b67b5e4d8","_number":768858,"virtual_id_number":768858,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-04 09:57:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2021-01-15 11:27:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-02 06:27:49.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 07:05:03.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 07:10:40.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-01-02 07:11:32.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 07:49:42.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 07:50:50.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-01-02 08:11:34.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 08:11:39.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 08:18:52.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-01-02 08:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-02 09:30:22.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-01-02 09:34:39.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-01-02 09:35:03.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 09:43:51.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 10:52:23.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-01-02 10:53:44.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 10:59:10.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-01-02 11:16:24.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-01-02 11:35:18.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 12:27:57.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-02 14:02:32.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"CC"},{"updated":"2021-01-04 10:49:39.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-04 13:21:49.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-01-15 00:57:32.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-15 00:57:32.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-01-15 00:57:32.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"6f9fbf391d6b92587e0d51cfd3f5d28b413ff62a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-02 06:27:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1849149762c194fb2e9b0c3cade93f9574950cd","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-02 06:27:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8534975c6d6664c061492581bf979fa0771c9263","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-02 07:05:03.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768858/768858/1/check-not-vote/ext-cinder-zuul/a741e1d : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"201ba9cba838b579dac7a3cd259e0efd623516d2","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-02 07:10:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768858/1/iscsi/ : FAILURE in 42m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/768858/1/smb/ : FAILURE in 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e447d6203896cd0e4f592344102fdb4d735f5ee5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-02 07:11:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/58/768858/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"b3ecfa00c063dc2395360e28c647666cd847dac4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-02 07:49:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/768858/1/ : FAILURE in 1h 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f6d27b79d2d7859535e566e99411af6c0ee76b18","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-02 07:50:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/768858/1/check/dsvm-tempest-cisco-zonemanager-job/cf3e2a2/ : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a115e5aa839464b24aa2f621a3094d830157587b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-02 08:02:51.000000000","message":"Topic set to bug/1909630","accounts_in_message":[],"_revision_number":1},{"id":"41444f897e2efe9cd432e0ba6f96ed44afb40836","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-02 08:11:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/768858/1/check/tempest-dsvm-full-focal-py3/b37d3f0/ : SUCCESS in 1h 30m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/768858/1/check/devstack-plugin-ceph-dsvm-focal-py3/c531af8/ : SUCCESS in 1h 35m 42s","accounts_in_message":[],"_revision_number":1},{"id":"58fbe08001fa2b78cb2886da4facedb4f95e6ce2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-02 08:11:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/768858/1/check/DellEMC_XtremIO_FC/9e523bc/DellEMC_XtremIO_FC/409 : SUCCESS in 1h 43m 43s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/768858/1/check/DellEMC_XtremIO_iSCSI/149e48e/DellEMC_XtremIO_iSCSI/395 : SUCCESS in 1h 35m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"b50a95b9461a95a45c6b016f46bcb354d139cec1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-02 08:18:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/58/768858/1/check/DellEMC_Unity_FC/8dd4d00/DellEMC_Unity_FC/395 : FAILURE in 1h 50m 56s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/58/768858/1/check/DellEMC_Unity_iSCSI/9a32450/DellEMC_Unity_iSCSI/370 : FAILURE in 1h 23m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"12d3287baa207a6706ef9ecff85ce375a851ec65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 08:39:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c4430e779ec446286f649782e568ca9 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1109bf70795e468b89bba5232ef25cb3 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4658ca1b1ee74527aa0a6fdcec417b78 : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2862f890ad724a27b6dc2147054a8dd5 : SUCCESS in 9m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97af0377ee3e429182a0861a23b99a0b : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6bdd3e4a4fe44b5b66b6ef10c658425 : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/725feaa4251241bcbb55ab367d336e06 : SUCCESS in 1h 04m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39e0a57fd2ca452297aa1f83c860149b : SUCCESS in 1h 37m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1db744c1b6a34b26b44392b731b48df2 : SUCCESS in 13m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1fc2b6ec3b644bfaa3582966df596eeb : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb8c970670374343a3457a4e8925b641 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/648b4eff4de54405b2f712c97f1c605d : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d619a6ef7ef54eb3844bd8edc12c669c : SUCCESS in 40m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/43b57ee71ede4bad9263df5a8d43f0c5 : FAILURE in 7m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d95865071c884d959c4bea06c320c054 : SUCCESS in 1h 05m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6cbd5325e02841cdabac15c1a0ff1f24 : FAILURE in 1h 18m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eeb316f96bed458ca01f723bc33170b6 : SUCCESS in 1h 42m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c6cd600bfef5425fa6e04da6d4c1ced4 : SUCCESS in 1h 09m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e8c67c92d2af41a0aec7fdf2cca235ad : SUCCESS in 1h 05m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc35a151bc574a3bb4206d765f810ac4 : SUCCESS in 1h 36m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/beb4e1485d3947c0bbfeb0f2a3e89836 : SUCCESS in 1h 55m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb6d7958562a4544992d0390580dcf2a : SUCCESS in 43m 35s","accounts_in_message":[],"_revision_number":1},{"id":"06b584d9247dadf6f7af165a310be85bb432befa","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-02 09:30:22.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/768858/1/check/DellEMC_PowerStore_iSCSI/296c073/DellEMC_PowerStore_iSCSI/220 : SUCCESS in 50m 59s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/768858/1/check/DellEMC_PowerStore_FC/16360fa/DellEMC_PowerStore_FC/214 : SUCCESS in 50m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"9168230746b4a5727a09708df0c7c9003f52a03e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-02 09:34:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/768858/1/upstream-solidfire-check/cinder-solidfire-iSCSI/48c080b/ : SUCCESS in 54m 14s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"3ac86c2b05c72b97727005bc332f9e076ffdc29b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-02 09:35:03.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-768858-1/ns5_nfs : SUCCESS in 2h 57m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-768858-1/ns5_iscsi : SUCCESS in 2h 55m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"e1293547e4b96a01669521689ab0e3aef12acb56","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-02 09:40:24.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/58/768858/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"5c387f58bbb509122f677c008fec9556f3030912","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-02 09:43:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/58/768858/1/upstream-check/cinder-cDOT-iSCSI/ae891fa/ : SUCCESS in 37m 09s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/58/768858/1/upstream-check/cinder-cDOT-iSCSI-IPv6/d765edd/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/58/768858/1/upstream-check/cinder-cDOT-NFS/c6a7776/ : SUCCESS in 37m 02s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/58/768858/1/upstream-check/cinder-cDOT-NFS-IPv6/3f2dd5d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/58/768858/1/upstream-check/cinder-cDOT-FCP/f1beafd/ : SUCCESS in 1h 03m 46s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e3c30b55228dbab8e98c7b1283128006b8774855","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-02 10:21:38.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/768858/1/check/DellEMC_XtremIO_FC/fcb523f/DellEMC_XtremIO_FC/410 : SUCCESS in 1h 42m 20s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/768858/1/check/DellEMC_XtremIO_iSCSI/875e964/DellEMC_XtremIO_iSCSI/396 : SUCCESS in 1h 36m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"0e2d3b1db526bf0f03801406740141ef15ed783b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-02 10:24:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/58/768858/1/check/DellEMC_Unity_FC/3f80ee0/DellEMC_Unity_FC/396 : FAILURE in 1h 45m 28s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/58/768858/1/check/DellEMC_Unity_iSCSI/e22d8e5/DellEMC_Unity_iSCSI/371 : FAILURE in 1h 26m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"968133d972fd7fab39339720fb325d4970a925e5","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-01-02 10:52:23.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/768858-1/PVME-iSCSI : SUCCESS in 52:21.03\n* PVME-FC http://os-logs.tristero.net/dell-ci/768858-1/PVME-FC : SUCCESS in 49:10.92","accounts_in_message":[],"_revision_number":1},{"id":"671930d10dede0bcc448ef708df088d1c8e2d9dd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-02 10:53:44.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/58/768858/1/check/DellEMC_PowerFlex/a592c91/DellEMC_PowerFlex/178 : SUCCESS in 2h 14m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"5ffb08ca2a1b709af16ed5e1c6b2693a290d2b2e","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-01-02 10:59:10.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/768858-1/HPMSA-iSCSI : FAILURE in 1:06:31\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/768858-1/HPMSA-FC : SUCCESS in 38:04.63","accounts_in_message":[],"_revision_number":1},{"id":"a4cd550309db09a01c039073a61a1b816fe61914","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-02 11:16:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/58/768858/1/check/DellEMC_SC_FC/573e47f/DellEMC_SC_FC/166 : FAILURE in 20s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/58/768858/1/check/DellEMC_SC_iSCSI/3d9546d/DellEMC_SC_iSCSI/169 : FAILURE in 2h 37m 06s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"47439c7d0a7ada6a38473c3f8ec81161e21f981e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-02 11:35:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/58/768858/1/check/DellEMC_VNX_iSCSI/04656df/DellEMC_VNX_iSCSI/221 : SUCCESS in 2h 55m 59s\n- DellEMC_VNX_FC http://publiclogs.emc.com/58/768858/1/check/DellEMC_VNX_FC/43b37e6/DellEMC_VNX_FC/222 : SUCCESS in 1h 21m 06s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"07a57575906ec52608657af27cbb2fab36dcce5a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-02 11:38:48.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/58/768858/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6adf2d3ce121a01fd41b6f77c5e8a92afd907fcd","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-02 12:27:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/58/768858/1/check/DellEMC_PowerMAX_FC/c82c5ef/DellEMC_PowerMAX_FC/330 : SUCCESS in 4h 02m 19s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/58/768858/1/check/DellEMC_PowerMAX_iSCSI/4752cdf/DellEMC_PowerMAX_iSCSI/303 : SUCCESS in 5h 17m 52s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"80ba0e825bfce3838eb4ecdc2de5953e7c27596a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-02 13:45:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/58/768858/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0e2e5560243e5bd7c684c9bddfd583a5094cc57d","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-01-02 14:02:32.000000000","message":"Patch Set 1:\n\nI\u0027m curious if we can add some unit tests for this fix","accounts_in_message":[],"_revision_number":1},{"id":"3b3a17b8fb60955cf5249b398a555886312e167c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-03 15:32:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/768858/1/ : RETRY_LIMIT in 24s","accounts_in_message":[],"_revision_number":1},{"id":"4226aaafb5b746bfb7748599b42cad93983e1771","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-03 18:29:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/768858/1/ : FAILURE in 2h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e9096875266e45d13998c682e693f6e7c196c0d5","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-04 05:47:36.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I\u0027m curious if we can add some unit tests for this fix\n\nhi, Ivan Kolodyazhny.\nthanks for your review, i have add unit tests about this fix.","accounts_in_message":[],"_revision_number":1},{"id":"d9bc3ddea7815cbe363d23c31ae62457175f1648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-04 05:48:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8272e72973b43c0d00f05fd38c8eb277977fdcbd","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-04 05:49:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1da9ce48b5b0b484d9208a44471a2c6ff73f6ee0","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-04 06:31:11.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768858/2/iscsi/ : FAILURE in 41m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/768858/2/smb/ : FAILURE in 33m 49s","accounts_in_message":[],"_revision_number":2},{"id":"99c9fc55aa0d37b81f00524621a044013f760b32","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-04 06:31:17.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/768858/768858/2/check-not-vote/ext-cinder-zuul/c4fa665 : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":2},{"id":"fa2d60c5c20500548ab733cb7f0fac7f2902d6c0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-04 06:50:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/768858/2/ : FAILURE in 55m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d8abc6ad0b6610eb32dee9433a440ff42859f9ef","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-04 07:11:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/768858/2/check/dsvm-tempest-cisco-zonemanager-job/8e332f9/ : SUCCESS in 1h 17m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a4486956d30b896d6a88ffcd31786d4a4cf0c201","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-04 07:26:51.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/768858/2/check/tempest-dsvm-full-focal-py3/45ea64b/ : FAILURE in 47m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/768858/2/check/devstack-plugin-ceph-dsvm-focal-py3/17844f9/ : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4c17ca923d52f7a2033311e7629eeab21c29b08f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-04 07:34:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"59631cc48949f58334a4881bfb1d29765c0f581d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-04 07:34:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5fb7a60f2a2353414e5cc34032356684ad057e2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-04 07:57:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0b9f9eefaa59c8df1803134bf1a13c978a39eb57","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-04 07:58:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9dc98ab86be36f0005c6fc384fb4e50bcc9f5021","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-04 08:03:33.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/768858/768858/4/check-not-vote/ext-cinder-zuul/b7d2580 : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":4},{"id":"4ea8df4d92a976f72bc0905b5b6c20cf81bc6936","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-04 08:39:58.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/768858/4/iscsi/ : FAILURE in 41m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/768858/4/smb/ : FAILURE in 33m 13s","accounts_in_message":[],"_revision_number":4},{"id":"a09d065b9272ad8314f82c1d1e1cb1b9bef68993","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-01-04 09:00:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/768858/4/ : FAILURE in 58m 08s","accounts_in_message":[],"_revision_number":4},{"id":"e13add06453f5e5e577a1949cbd4cabf6c605b9d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-04 09:21:46.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-768858-2/ns5_nfs : SUCCESS in 3h 22m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-768858-2/ns5_iscsi : SUCCESS in 3h 9m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5ea048b8bf41cffacf8a6324c736844c21e91513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 09:57:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89fb614b1cb8478c9a0b2ddc374f15df : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0383be9f20448a9c50ca0dda7900b2 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abce5db831414ee4af0495674ffb396b : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5daaa39eb5ae44a28c8a28a421db2651 : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fbcd2bb8f97409a9b064f719f20baac : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb710525720c4a9f802e6d77db5fd551 : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/ec40835af899497f8222d77618dfcd88 : SUCCESS in 1h 02m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b92123629e7345d68c3e60805cf2d508 : SUCCESS in 1h 04m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/05cef1abe54b41788604efe14d599a0f : SUCCESS in 10m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7eab710281894919852f928b7a0d4dc8 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45c74f46c951433d9662d424760e0307 : SUCCESS in 5m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1605387394174934813872c001325fea : SUCCESS in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0cfe6c0d7bf24bbebb31cdb4f64e086a : SUCCESS in 33m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fda2687fad3248409a36bd935649e255 : FAILURE in 5m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a966f65a3f2f4115b1cf1ae7926a54ba : SUCCESS in 57m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e71df6d1d5794b6dbb2194da93d2289a : FAILURE in 1h 17m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a6b29d367a649f6a6ddf9e64cb2c5e1 : SUCCESS in 1h 22m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/260725e7baad4409a1995bf0e1e92858 : SUCCESS in 1h 09m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d387b326bc114082a108a778a1124ce9 : SUCCESS in 1h 06m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/810aa40e268e413b95ecfa053f663f7e : SUCCESS in 1h 13m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d15cf76b8674cef9fd9b29f1e70d88d : SUCCESS in 1h 58m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9d120b5d53549a49e74d084574f2ddc : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":4},{"id":"19b1d2f7855a3c3fc49df0fda4a9b0380aa7b17f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-04 10:00:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/58/768858/4/check/dsvm-tempest-cisco-zonemanager-job/fd4fff3/ : SUCCESS in 1h 16m 40s","accounts_in_message":[],"_revision_number":4},{"id":"6bb75bb70fb177b967e3737da45615d751d266df","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 10:20:08.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/58/768858/4/check/DellEMC_XtremIO_FC/8587592/DellEMC_XtremIO_FC/439 : SUCCESS in 1h 40m 47s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/58/768858/4/check/DellEMC_XtremIO_iSCSI/30718b0/DellEMC_XtremIO_iSCSI/424 : SUCCESS in 1h 37m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"698f6f9db43242f8817f248192c81daa78d380b0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-04 10:31:30.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/768858/4/check/tempest-dsvm-full-focal-py3/c921d6d/ : SUCCESS in 1h 30m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/58/768858/4/check/devstack-plugin-ceph-dsvm-focal-py3/6aaf840/ : SUCCESS in 2h 26m 21s","accounts_in_message":[],"_revision_number":4},{"id":"3e017e7e442c27220f104cee7e6c0ee3281a7600","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-04 10:47:24.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-768858-3/ns5_nfs : SUCCESS in 3h 12m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-768858-3/ns5_iscsi : SUCCESS in 3h 8m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8fc5c2acbcbea032076fb32f6c46c5020904659c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-04 10:49:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/58/768858/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/107114e/107114e.html : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e18061b73a7cdb9363008ffd49c849c08fc26431","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-04 11:09:58.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-768858-4/ns5_nfs : SUCCESS in 3h 5m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-768858-4/ns5_iscsi : SUCCESS in 3h 11m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"958a594db9b8651b3f59a3b3a19e1cc61c6b1dc1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-04 11:21:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/768858/4/upstream-solidfire-check/cinder-solidfire-iSCSI/ceeeba7/ : SUCCESS in 1h 22m 23s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"a522e180a8b78578c045ad3d0a027627d2794b10","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-04 11:46:42.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/58/768858/4/check/DellEMC_PowerStore_iSCSI/1720428/DellEMC_PowerStore_iSCSI/223 : SUCCESS in 52m 09s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/58/768858/4/check/DellEMC_PowerStore_FC/44c26c3/DellEMC_PowerStore_FC/217 : SUCCESS in 54m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"61b388df69e3922697ee0cb27c3caa85f78794b9","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-04 13:21:49.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/768858-4/Seagate-iSCSI : SUCCESS in 46:16.38\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/768858-4/Seagate-FC : SUCCESS in 40:54.35","accounts_in_message":[],"_revision_number":4},{"id":"d9ac50fa38706c24f68e6cee95eb5f1aca0831d4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-04 13:23:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/58/768858/4/upstream-check/cinder-cDOT-iSCSI/8372141/ : SUCCESS in 36m 03s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/58/768858/4/upstream-check/cinder-cDOT-iSCSI-IPv6/6c4e2a3/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/58/768858/4/upstream-check/cinder-cDOT-NFS/d7b5a1c/ : SUCCESS in 36m 06s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/58/768858/4/upstream-check/cinder-cDOT-NFS-IPv6/4fef87d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/58/768858/4/upstream-check/cinder-cDOT-FCP/95616a7/ : FAILURE in 3h 25m 01s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"931f3a9c8b158bb3505cd66d22260abf97f7e987","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-01-04 13:40:45.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/768858-4/PVME-iSCSI : SUCCESS in 52:28.51\n* PVME-FC http://os-logs.tristero.net/dell-ci/768858-4/PVME-FC : SUCCESS in 49:16.37","accounts_in_message":[],"_revision_number":4},{"id":"9a6aea8999fc68f4957f8f92005579d0e60e13fd","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-01-04 13:52:38.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/768858-4/HPMSA-iSCSI : FAILURE in 1:09:23\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/768858-4/HPMSA-FC : SUCCESS in 37:36.62","accounts_in_message":[],"_revision_number":4},{"id":"434f5ebbd18c867062325e8e4b98de6b97a7f85a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 14:25:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/58/768858/4/check/DellEMC_PowerMAX_FC/d06c753/DellEMC_PowerMAX_FC/359 : SUCCESS in 4h 22m 36s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/58/768858/4/check/DellEMC_PowerMAX_iSCSI/ded6dc4/DellEMC_PowerMAX_iSCSI/330 : SUCCESS in 4h 54m 56s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"d131eca2d5be5322ad5349ceddb604c1c6ff5f39","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 14:34:56.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/58/768858/4/check/DellEMC_PowerFlex/8714cd6/DellEMC_PowerFlex/181 : SUCCESS in 2h 15m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"389c17fffb20dc07bc99be0b3a16166d58abf48b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 16:58:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/58/768858/4/check/DellEMC_VNX_iSCSI/089325f/DellEMC_VNX_iSCSI/225 : SUCCESS in 3h 05m 05s\n- DellEMC_VNX_FC http://publiclogs.emc.com/58/768858/4/check/DellEMC_VNX_FC/f3ceed5/DellEMC_VNX_FC/226 : FAILURE in 11m 18s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"59a0bc0a581c15899be063da6fc5b70ced1e8d99","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-01-15 11:27:00.000000000","message":"Patch Set 4: Code-Review-2\n\nSorry, I don\u0027t understand why you say that there is no quota_classes... We have the REST API to create them: https://docs.openstack.org/api-ref/block-storage/v3/#update-quota-classes-for-a-project\n\nI believe the code this patch is trying to remove is necessary, so I\u0027ll -2 the patch for the moment, but I\u0027ll change it if I\u0027m proved wrong.","accounts_in_message":[],"_revision_number":4},{"id":"5d5bbe383521d7fe7ca7b8e43684a8aec49dd3da","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-19 08:45:08.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review-2\n\u003e \n\u003e Sorry, I don\u0027t understand why you say that there is no quota_classes... We have the REST API to create them: https://docs.openstack.org/api-ref/block-storage/v3/#update-quota-classes-for-a-project\n\u003e \n\u003e I believe the code this patch is trying to remove is necessary, so I\u0027ll -2 the patch for the moment, but I\u0027ll change it if I\u0027m proved wrong.\n\nthanks for your review, Gorka Eguileor.\ni know we have rest to create some item in quota_class table, we can set default quota for all new project. but it doesn\u0027t make sense to set default quota for an volume type with specified name(\"%s_%s\" % (quota_type, volulme_type_name)), because we do not know the volume type name before we create it. this means \"%s_%s\" % (quota_type, volulme_type_name) will not appear in quota_class table. even if we set it in manually, it means nothing. so we not necessary try to update it when update volume type name.","accounts_in_message":[],"_revision_number":4},{"id":"80ebde2da233edcc2bfe74b59eff9f092ecc035d","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-01-19 11:01:39.000000000","message":"Patch Set 4:\n\nHi Haixin,\n\nThere may be a misunderstanding, as you mention that the name may not be known on creation, but this is called upon updating the name of a type.\n\nSo here\u0027s a case where this is necessary:\n\n- Create volume type\n- Update quota class for that type\n- Update the volume type name\n\nIf I\u0027m not mistaken, this patch removes the code that preserves the connection between the volume type and the quota class.  So without that update the volume type will have a new name and the quota class will no longer apply to it.","accounts_in_message":[],"_revision_number":4},{"id":"c3453ee51378e4bc5813965e4cc1290f3e06bafa","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-20 07:12:47.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e Hi Haixin,\n\u003e \n\u003e There may be a misunderstanding, as you mention that the name may not be known on creation, but this is called upon updating the name of a type.\n\u003e \n\u003e So here\u0027s a case where this is necessary:\n\u003e \n\u003e - Create volume type\n\u003e - Update quota class for that type\n\u003e - Update the volume type name\n\u003e \n\u003e If I\u0027m not mistaken, this patch removes the code that preserves the connection between the volume type and the quota class.  So without that update the volume type will have a new name and the quota class will no longer apply to it.\n\nHi, Gorka Eguileor.\n    As we know, quota_class is used to set default quota for all new project, but not an specified project, so i think it should be true for volume type. maybe we need another patch to fix this logic about this. after fixed, we can set default quota in quota_class for all new volume type, instead of for an specified volume type with type name.\n    After we create an volume type, if we want to set quota for it, we just need to update quota for that type, but not update quota class. \n    Default quota means we set default quota before we create resource, and all new created resource follow this quota value. rather than we can noly set default quota after we creat resource.\n    Any new volume type should follow default quota in quota class if we have set, otherwise quota is unlimited. if need to set quota after created, just to update quota table for it.\n   how do you think?","accounts_in_message":[],"_revision_number":4},{"id":"7792ea0753ded9a29d01cf654d2eda0b67b5e4d8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-08-16 07:27:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b6ff573707fe99cfc2af480cfcc61c4fcf47e1fc","revisions":{"6d59adc9a85f27043c87d77c598b84c9c83dda16":{"kind":"REWORK","_number":1,"created":"2021-01-02 06:27:35.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/58/768858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/768858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/768858/1"}}},"commit":{"parents":[{"commit":"8240195bd6433cc8d718db76878ea3d244580bc1","subject":"Merge \"RBD: Retry delete if VolumeIsBusy in _copy_image_to_volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8240195bd6433cc8d718db76878ea3d244580bc1"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-02 06:23:44.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-02 06:26:17.000000000","tz":480},"subject":"Remove invalid quota resource when update volume type name","message":"Remove invalid quota resource when update volume type name\n\nThere is no such resource named as \"%s_%s\" %\n(quota_type, volulme_type_name) in database table of quota_classes.\n,there are only quota default value for new project.\nUnless we manually set the default quota for volume type. These values\nare default values, and we don\u0027t know how the user will name the volume\ntype before we create it,So it doesn\u0027t make sense to preset a default\nvalue for a volume type in advance.\n\nso resource named as \"%s_%s\" % (quota_type, volulme_type_name) only\nexist in table quota_usages or quotas if we set quota for specified\nvolume type.\nwill not appear in table of quota_classes.\n\nso we not need to call quota_class_update_resource when update volume\ntype name.\n\nCloses-Bug: #1909630\nChange-Id: If395e63bd2cb22b5bc3f807755631b81ad3ac015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d59adc9a85f27043c87d77c598b84c9c83dda16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d59adc9a85f27043c87d77c598b84c9c83dda16"}]},"branch":"refs/heads/master"},"1f09a7bf992bfff826262e21ce80f46a719c4dc6":{"kind":"REWORK","_number":2,"created":"2021-01-04 05:48:57.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/58/768858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/768858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/768858/2"}}},"commit":{"parents":[{"commit":"8240195bd6433cc8d718db76878ea3d244580bc1","subject":"Merge \"RBD: Retry delete if VolumeIsBusy in _copy_image_to_volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8240195bd6433cc8d718db76878ea3d244580bc1"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-02 06:23:44.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-04 05:48:22.000000000","tz":480},"subject":"Remove invalid quota resource when update volume type name","message":"Remove invalid quota resource when update volume type name\n\nThere is no such resource named as \"%s_%s\" %\n(quota_type, volulme_type_name) in database table of quota_classes.\n,there are only quota default value for new project.\nUnless we manually set the default quota for volume type. These values\nare default values, and we don\u0027t know how the user will name the volume\ntype before we create it,So it doesn\u0027t make sense to preset a default\nvalue for a volume type in advance.\n\nso resource named as \"%s_%s\" % (quota_type, volulme_type_name) only\nexist in table quota_usages or quotas if we set quota for specified\nvolume type.\nwill not appear in table of quota_classes.\n\nso we not need to call quota_class_update_resource when update volume\ntype name.\n\nCloses-Bug: #1909630\nChange-Id: If395e63bd2cb22b5bc3f807755631b81ad3ac015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f09a7bf992bfff826262e21ce80f46a719c4dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f09a7bf992bfff826262e21ce80f46a719c4dc6"}]},"branch":"refs/heads/master"},"3c390697912e02a8556d805d4a7db4a0d513573c":{"kind":"REWORK","_number":3,"created":"2021-01-04 07:34:19.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/58/768858/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/768858/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/768858/3"}}},"commit":{"parents":[{"commit":"8240195bd6433cc8d718db76878ea3d244580bc1","subject":"Merge \"RBD: Retry delete if VolumeIsBusy in _copy_image_to_volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8240195bd6433cc8d718db76878ea3d244580bc1"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-02 06:23:44.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-04 07:33:45.000000000","tz":480},"subject":"Remove invalid quota resource when update volume type name","message":"Remove invalid quota resource when update volume type name\n\nThere is no such resource named as \"%s_%s\" %\n(quota_type, volulme_type_name) in database table of quota_classes.\n,there are only quota default value for new project.\nUnless we manually set the default quota for volume type. These values\nare default values, and we don\u0027t know how the user will name the volume\ntype before we create it,So it doesn\u0027t make sense to preset a default\nvalue for a volume type in advance.\n\nso resource named as \"%s_%s\" % (quota_type, volulme_type_name) only\nexist in table quota_usages or quotas if we set quota for specified\nvolume type.\nwill not appear in table of quota_classes.\n\nso we not need to call quota_class_update_resource when update volume\ntype name.\n\nCloses-Bug: #1909630\nChange-Id: If395e63bd2cb22b5bc3f807755631b81ad3ac015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c390697912e02a8556d805d4a7db4a0d513573c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c390697912e02a8556d805d4a7db4a0d513573c"}]},"branch":"refs/heads/master"},"b6ff573707fe99cfc2af480cfcc61c4fcf47e1fc":{"kind":"REWORK","_number":4,"created":"2021-01-04 07:57:46.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/58/768858/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/768858/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/768858/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/768858/4"}}},"commit":{"parents":[{"commit":"8240195bd6433cc8d718db76878ea3d244580bc1","subject":"Merge \"RBD: Retry delete if VolumeIsBusy in _copy_image_to_volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8240195bd6433cc8d718db76878ea3d244580bc1"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-02 06:23:44.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-04 07:57:19.000000000","tz":480},"subject":"Remove invalid quota resource when update volume type name","message":"Remove invalid quota resource when update volume type name\n\nThere is no such resource named as \"%s_%s\" %\n(quota_type, volulme_type_name) in database table of quota_classes.\n,there are only quota default value for new project.\nUnless we manually set the default quota for volume type. These values\nare default values, and we don\u0027t know how the user will name the volume\ntype before we create it,So it doesn\u0027t make sense to preset a default\nvalue for a volume type in advance.\n\nso resource named as \"%s_%s\" % (quota_type, volulme_type_name) only\nexist in table quota_usages or quotas if we set quota for specified\nvolume type.\nwill not appear in table of quota_classes.\n\nso we not need to call quota_class_update_resource when update volume\ntype name.\n\nCloses-Bug: #1909630\nChange-Id: If395e63bd2cb22b5bc3f807755631b81ad3ac015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6ff573707fe99cfc2af480cfcc61c4fcf47e1fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6ff573707fe99cfc2af480cfcc61c4fcf47e1fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
