)]}'
{"id":"openstack%2Fcinder~769546","triplet_id":"openstack%2Fcinder~master~Ie04d6dbdb9d18047d9167f85b4346d15babca4f3","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"Ie04d6dbdb9d18047d9167f85b4346d15babca4f3","subject":"Remove __unicode__() from CinderException","status":"MERGED","created":"2021-01-06 14:38:47.000000000","updated":"2021-02-14 06:08:59.000000000","submitted":"2021-02-11 03:23:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769546","meta_rev_id":"b058b5e5f9cfca41f56a708293d954df3187aac1","_number":769546,"virtual_id_number":769546,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-11 03:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},"all":[{"value":1,"date":"2021-01-06 14:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},{"value":2,"date":"2021-02-10 17:20:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-01-13 07:16:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-07 15:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},{"value":1,"date":"2021-02-10 17:20:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-06 14:39:06.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 14:51:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 14:56:22.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 14:58:22.000000000","updated_by":{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},"reviewer":{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},"state":"REVIEWER"},{"updated":"2021-01-06 15:46:35.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-01-06 16:10:02.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-01-06 16:28:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 17:00:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 17:05:27.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 17:17:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-06 18:28:53.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 19:29:35.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 22:01:34.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-01-06 22:42:04.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 23:02:00.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-01-06 23:14:53.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-01-06 23:36:10.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-01-07 00:04:50.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-07 06:20:32.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-07 15:47:37.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-01-13 07:16:57.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-02-10 17:20:30.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-11 14:03:46.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-02-11 15:02:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-11 15:31:57.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-11 17:54:03.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-02-14 04:52:15.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"00358d14589e017c10c059f297b73d942291ff48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-01-06 14:38:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dad6fdb61a7bccc6b9b4400ce742397cf6946e6","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-06 14:39:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f98c7edd74ea950203129aa52797529296f31f5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-01-06 14:51:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/46/769546/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d9ad470/d9ad470.html : FAILURE in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1c80f26d93d23cc6b1461e1c2135ec59f33b1c1e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-06 14:56:22.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-46-769546-1/ns5_nfs : FAILURE in 16m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-46-769546-1/ns5_iscsi : FAILURE in 16m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f6cbee88481ee78666441712ba21d23ab482b909","author":{"_account_id":23601,"name":"michael","display_name":"michael-mcaleer","email":"Michael.Mcaleer@dell.com","username":"michael-mcaleer"},"date":"2021-01-06 14:58:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"df0b89d65a9b7ff4c199961b5504addb4295a6df","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-06 15:46:35.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/769546/1/iscsi/ : SUCCESS in 1h 06m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/769546/1/smb/ : FAILURE in 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"74a40c0d38baadd058c89c05b974f43ba98573d2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-06 16:10:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/46/769546/1/check/dsvm-tempest-cisco-zonemanager-job/fb1fdc7/ : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":1},{"id":"199d56ee45fdd7e47270915f25b4dbb629fb9a9d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-06 16:28:48.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/769546/769546/1/check-not-vote/ext-cinder-zuul/eea3c31 : SUCCESS in 1h 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0a1d5adc9118e53bb2767e92779b8c9f2dd3238f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-06 17:00:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/46/769546/1/check/tempest-dsvm-full-focal-py3/29f22e2/ : SUCCESS in 1h 36m 36s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/46/769546/1/check/devstack-plugin-ceph-dsvm-focal-py3/269cfd0/ : SUCCESS in 1h 58m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6a8a8586796c43def16d9efd45914f638c608af2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-06 17:05:27.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_XtremIO_FC/5823a6e/DellEMC_XtremIO_FC/458 : SUCCESS in 2h 05m 09s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_XtremIO_iSCSI/6af7a42/DellEMC_XtremIO_iSCSI/445 : SUCCESS in 1h 28m 41s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"e12a7b9317b0d34d743187b0e0697a5c6ee7e7d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 17:17:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b06e75fc00eb47bdbc3aab84e824bb32 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f702fba56f04156aafd303d7cb12530 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b9c7de3601c4ba4bb7747abcf496a13 : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f81074b80324fb58637bcb68b2f22c2 : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a163a3d8c1a4e3289bf2e48027a5ece : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c8756b6c3047e29c5d9beb3b94139f : SUCCESS in 11m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/3d7eacb1371141db9e9577f61c9e72f8 : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/872edf4cbed14d10ad01fd2bae92a23d : SUCCESS in 1h 15m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c8d04efc92e64edc90adad831a0857e6 : SUCCESS in 11m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af37eeaa4f2547b8ab8196353cea9dd3 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a431d46cac24a809642473217d581b9 : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6fae978561cb415d81c21ff9e9835d74 : SUCCESS in 5m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d045ea24065248f8b4aed919599cff6b : SUCCESS in 41m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54afe0422f274a72a749ec273631f6be : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea12e5b456be494dba8795884263c787 : SUCCESS in 1h 18m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9f8a0aef9e45477c9d0afacd69eacc09 : FAILURE in 1h 18m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/851b4820835f493cbc38f7feb4d95ab6 : FAILURE in 1h 39m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/af983eda3f23471daa12627d6246f483 : SUCCESS in 1h 34m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/90c9df17f44a44d3a8e837abd0185514 : SUCCESS in 1h 28m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6b2e0882c61e4bb083414670cdb1bfa0 : FAILURE in 2h 25m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5a3f8a15aa94f22a7ef020627afe797 : SUCCESS in 2h 16m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0315205adb5a4698bee3495ceca98a72 : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"868fc4f4444e436e191d65882d11203939a78a17","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-01-06 18:28:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/46/769546/1/20210107012400/NEC-ISCSI : SUCCESS in 1h 58m 54s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/46/769546/1/20210107012400/NEC-FC : SUCCESS in 2h 04m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7814a10fb61a41eac562b97fadda52aa261472cf","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-01-06 18:45:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c84a601b6644a0e7c2279e107881082b6ee983e7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-06 18:45:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a613e67330168e79c471bbc7e9f26b6444e297a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-06 19:29:35.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"399385f64eef8ad4fccadff7b5b7746a663dbd6d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-01-06 19:50:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/769546/1/iscsi/ : SUCCESS in 1h 03m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/769546/1/smb/ : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":1},{"id":"210728ecca31f9dd786703aef012ed6284ef5e33","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-06 20:03:13.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/769546/769546/1/check-not-vote/ext-cinder-zuul/4bbbf3d : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1c2ad3cea31fb1029b27c3142635126c988e91c1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-06 20:58:55.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-46-769546-1/ns5_nfs : FAILURE in 4h 8m 52s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-46-769546-1/ns5_iscsi : FAILURE in 3h 33m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"fb44a832fd4d3a3c629a142e101594c6a0a6e830","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-01-06 20:59:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/46/769546/1/check/dsvm-tempest-cisco-zonemanager-job/0028c77/ : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c0edf47bb895714932d3559fe0416a8742e4cd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 21:09:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dc657f251d54b6ba87337572122dd4b : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb90f0f5bcc4b82b60d917b6b9bd42a : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b2b8e884c994f628b54ff266acc01dc : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91e6c37215174ff3ac1a3004411e55c2 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a76648ce82b24aa58eb93de44c08e61b : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f83bb6c3a14c37bc28e9b1ea09c521 : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/5326aed39ba4472b95058fbfa09fa418 : SUCCESS in 57m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2cd648f9d3df4569a1d46e51f01de277 : SUCCESS in 1h 08m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/83516f6b17bb4adba3f5a0877c908a18 : SUCCESS in 17m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a33fcd74a5a4f98ae19076ed1b4a810 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7fe5af767f954be6aa96507fa54002d8 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c007fbd13ce74bcbb2c9c4e35363df97 : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce0a1df50a49430ab2d51e5e09ec01f4 : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4297e56fec254bef8963e703e993566a : SUCCESS in 5m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32242b615b9f4a80aa7852d46fe49b4f : SUCCESS in 1h 12m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bd1f577fd2184a98ae679b992022202c : FAILURE in 1h 18m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/41292bf32bf5459e8e0906a7adf25982 : SUCCESS in 1h 31m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e1b93823ce4467d90d6605b14020ccf : SUCCESS in 1h 14m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b7ae3a32575b4feebfd0cff82eed717b : SUCCESS in 1h 24m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5cf3935a2404d088bcb99ca86235708 : SUCCESS in 1h 17m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5a77d8d7c224594b67044380f25fcd6 : SUCCESS in 2h 09m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4150c6ba41324c5ea37c9f821da7481a : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5824a5bbf17f921f39ac98c8e6479dd3272a7b53","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-06 22:01:34.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_PowerStore_iSCSI/f0ee83c/DellEMC_PowerStore_iSCSI/240 : SUCCESS in 52m 09s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_PowerStore_FC/8ab82d5/DellEMC_PowerStore_FC/234 : SUCCESS in 51m 05s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"2d222e765429cab874c291982dcbe9c2332776fc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-06 22:42:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-iSCSI/6c2cf7e/ : SUCCESS in 41m 47s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-iSCSI-IPv6/547655d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-NFS/ac4a329/ : SUCCESS in 40m 33s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-NFS-IPv6/4d7ea12/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-FCP/9bc064c/ : SUCCESS in 1h 31m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d9b5186e9c17933f162786b2a6b33096070582b0","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-06 23:02:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/46/769546/1/upstream-solidfire-check/cinder-solidfire-iSCSI/ce54e27/ : SUCCESS in 1h 51m 15s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"4db426a31f289d0209dc3b5ba429f735642759d0","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-01-06 23:14:53.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/769546-1/Seagate-iSCSI : SUCCESS in 46:03.61\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/769546-1/Seagate-FC : SUCCESS in 44:21.79","accounts_in_message":[],"_revision_number":1},{"id":"c7b758faba3c911d4368d02e0143540589a16440","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-06 23:25:49.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/46/769546/1/check/tempest-dsvm-full-focal-py3/57b40df/ : SUCCESS in 1h 44m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/46/769546/1/check/devstack-plugin-ceph-dsvm-focal-py3/722c554/ : TIMED_OUT in 2h 39m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8185798488824dff4d0abdc4108425961e104389","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-01-06 23:36:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_SC_FC/ca57636/DellEMC_SC_FC/186 : SUCCESS in 2h 05m 04s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_SC_iSCSI/9cf4d24/DellEMC_SC_iSCSI/186 : SUCCESS in 1h 30m 05s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f965c94fbe189ffb459c93913d7a868ed70c542","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-07 00:04:50.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/46/769546/1/check/DellEMC_PowerFlex/b1dde72/DellEMC_PowerFlex/198 : SUCCESS in 2h 15m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"c353a3c84381b5cd193c25400e45cef8764eb884","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-07 06:20:32.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_VNX_iSCSI/91b3be1/DellEMC_VNX_iSCSI/242 : SUCCESS in 2h 54m 44s\n- DellEMC_VNX_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_VNX_FC/510829b/DellEMC_VNX_FC/243 : SUCCESS in 1h 00m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"3aa7d510582b56705a742930d503c80b95b3c6ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-07 08:48:36.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"637b8f0c06c150dedbce0307af30d69519734ca7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-07 09:47:19.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_XtremIO_FC/e1d2e56/DellEMC_XtremIO_FC/466 : SUCCESS in 2h 04m 25s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_XtremIO_iSCSI/386fe96/DellEMC_XtremIO_iSCSI/453 : SUCCESS in 1h 31m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"53c55107ec6a6a4b8f83d6a9ad4fdeb3c583bec7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-07 14:26:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3a4cc153751f41f68cf9b67df9f537125d1818b9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-07 15:47:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c91b29b63edb5592b3b18b76350219a155041c32","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-12 00:34:19.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-46-769546-1/ns5_nfs : FAILURE in 5h 22m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-46-769546-1/ns5_iscsi : FAILURE in 4h 3m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"294decc67b706eb43aedb3f7742c31ab1ee50437","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-12 05:29:35.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4f663d2432777d57e80cbf280e8ec3ab19d15ebc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-12 07:22:35.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-46-769546-1/ns5_nfs : SUCCESS in 3h 53m 27s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-46-769546-1/ns5_iscsi : SUCCESS in 3h 50m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7a7c50ddb4a5375e79f261d76f48f9f842aabe54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-12 14:27:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"369087544adc33cc62c2c47bced783343698f53a","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-01-13 07:16:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a14e814de1a2badbc01ded845bbb6f7d075685ae","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-10 17:20:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM.","accounts_in_message":[],"_revision_number":1},{"id":"4ef60f88a1a1750b4afd5be18ceca4b5fdc76ed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 17:24:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09afd4a34468aaad9af89a0e112d127bfa06be43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 19:07:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07df501aca3b4fd1b3ea1162515bb994 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3500f2440b8c474396d033b53a37c053 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cddbb03c981420bbda1613a55291c2f : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e266f47927974648b8d75b9f3833893a : SUCCESS in 9m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26e8e3d6f584d6a965f5d9a5d66af76 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/4becf1624fae410a94fae38db09f6c6a : SUCCESS in 1h 05m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30789c27d99c40b8a06f4613eccb2cdb : SUCCESS in 1h 14m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf6eb2648001421998d8e3781f5cfd75 : SUCCESS in 1h 18m 22s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e359bb407bd94f7680975b1020cd10c8 : FAILURE in 24m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/922900296860422ca0e782939fdef45e : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9d12a86149ab41796cf05d8e715367c6d3ca3d88","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-10 22:51:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bdd8db4c3add63e923f4027bff6316a859402862","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-10 22:51:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dded9438178985c0dd6cc25660136e75d79b6d7e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-10 23:34:04.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/769546/769546/1/check-not-vote/ext-cinder-zuul/74700c3 : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2467eb5d5618c0a77016d8f02684477e54a59dae","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-11 00:35:10.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/769546/1/iscsi/ : FAILURE in 1h 43m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/769546/1/smb/ : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c6fe49d195cdc88676bed8c8c4397c95023bd89a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 01:07:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd6f67fa112b47358393f10fd741f39d : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd55275b0fd045a0b2b0bc5b846aa9a9 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eba26c1462494371aa6f9f1f05e32ed0 : SUCCESS in 11m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/115286c5c93949fcaff993b7a135dd56 : SUCCESS in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/619f348c9f33453aa50ffa40d7c522c5 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4165798d7c34df6ad363e2415ecb0fd : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/9fa5bc38f413445c86b36784f9d81952 : SUCCESS in 58m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ec73d9e8c9a42a5b4d17437cfb95a12 : SUCCESS in 1h 13m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/56870b43871e4319a7aadf7ec76af29d : SUCCESS in 12m 10s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb52e56eada44c55a09df866642237f7 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d4a3fafbe9684ce08b922b5ef4e769b5 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a874231ce0b44e108fd031583ca658f0 : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/db9ba5f3c85a4940b67e4afef361e246 : SUCCESS in 40m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20a054a331e542f89cf471ef58c6f95e : SUCCESS in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63e9b13892234ee48b77fc119dce1d7b : SUCCESS in 1h 08m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6499ae1aa9d746139c1701004b7088af : FAILURE in 1h 19m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a7abc6bccc74985b4e086e57450f157 : SUCCESS in 1h 31m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a89a2c140f4743e785c1661f9cdf3516 : SUCCESS in 1h 12m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fc5151c5c0214a48b453b1987c93b40c : SUCCESS in 1h 21m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cf92435841f94e669a7a0b47e01b2893 : SUCCESS in 1h 29m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a6951f4f67624625a0f10b23eb13ed93 : SUCCESS in 2h 05m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8af87183f70d4a1d9e4b63e520546920 : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ba3fbe913b118a7fd02660722639b2fa1abc4cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 01:07:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e91333ab237100890e78cdda73d1a2a45be34993","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-11 01:57:46.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7a49f2ab75189c281b06823d0355833af134368c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-11 02:00:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/46/769546/1/upstream-solidfire-check/cinder-solidfire-iSCSI/5009dd0/ : SUCCESS in 52m 49s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"ebaa53c59a3f6ab27d9e49123f5f483b7a513656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 03:23:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef8dcd2572b14c8198a8c9a272041da0 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/375edd44fadc45969aa33b5f50053e21 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d33c4981c7aa42178a477a23b395c23b : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2b7fc536c9f49dda53af3addf87473e : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76cae7addfa6415082b56c2314427259 : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/555701697ab1419bae886dce8a12067b : SUCCESS in 1h 02m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/caa3999c22c44e2b9dcf48d5ed3e9859 : SUCCESS in 1h 03m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/885217d7cc794d948ef226bb9b387cc7 : SUCCESS in 1h 08m 20s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/062c60a6ed0547d8984e099e73f6af2b : SUCCESS in 1h 06m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33427a72289d4a6088fca79404fdc7c4 : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7d2e23218c717147c8731caea9fd1089f661547d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 03:23:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3b844a5514a60c0d1a0ea7720f04de41f56a2fe6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 03:26:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/190e60a97d7a4e139106eeeafbe5a06e : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d37c73786d5f0eebc9d8ad1c901c2ea5f2292909","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-11 03:46:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-iSCSI/5462722/ : SUCCESS in 37m 29s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-iSCSI-IPv6/1cf848f/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-NFS/8e5a8a9/ : SUCCESS in 37m 02s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-NFS-IPv6/0edf4ba/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/46/769546/1/upstream-check/cinder-cDOT-FCP/5e2c991/ : FAILURE in 1h 40m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4ff195c5019a29802b7d0a0f72a62cf51fd3702d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-11 04:00:04.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_PowerStore_iSCSI/46a69cf/DellEMC_PowerStore_iSCSI/385 : SUCCESS in 54m 50s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_PowerStore_FC/c702cca/DellEMC_PowerStore_FC/379 : SUCCESS in 48m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"c694a2d1ddf5ba64124d9622e51e562356f34b5f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-11 04:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/46/769546/1/check/dsvm-tempest-cisco-zonemanager-job/1adca46/ : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"28274746a8cffea186eafc513a0b379a7e527583","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-11 05:36:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"79124a7e3eccfe1b662c05f1cf490ac361bb6de0","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-11 14:03:46.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/769546-1/Lenovo-iSCSI : SUCCESS in 49:49.25\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/769546-1/Lenovo-FC : SUCCESS in 47:25.90","accounts_in_message":[],"_revision_number":1},{"id":"f5c668b96cae64eaf435755d8e50d7c06b9fd4dd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-11 14:15:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/46/769546/1/check/DellEMC_PowerFlex/7bc484f/DellEMC_PowerFlex/368 : FAILURE in 2h 35m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"46a32fac0b95f223d4feb03eb4233178a4804237","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-11 14:29:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_SC_FC/e83c047/DellEMC_SC_FC/314 : FAILURE in 2h 16m 08s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_SC_iSCSI/9ad2704/DellEMC_SC_iSCSI/314 : SUCCESS in 2h 00m 09s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"25bce2e6b22d7aa9acba07a122509873e0221948","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-11 14:45:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3418a29ef7df5599e00601ae9b75c7db4db71b79","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-11 15:02:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_Unity_FC/cfbb90b/DellEMC_Unity_FC/669 : SUCCESS in 1h 38m 41s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_Unity_iSCSI/3436dbf/DellEMC_Unity_iSCSI/631 : SUCCESS in 1h 47m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"30c9005a3849abc3f925f5036e29f60f48a9020f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-11 15:08:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e0f564c044953f8089f533571b4cf7a315fbea3a","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-11 15:31:57.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/769546-1/PVME-iSCSI : SUCCESS in 50:18.31\n* PVME-FC http://os-logs.tristero.net/dell-ci/769546-1/PVME-FC : SUCCESS in 49:39.98","accounts_in_message":[],"_revision_number":1},{"id":"c40e581e846120bc7167c5e05657c4d05447dacf","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-11 16:17:40.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_VNX_iSCSI/6ef2427/DellEMC_VNX_iSCSI/368 : SUCCESS in 2h 58m 58s\n- DellEMC_VNX_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_VNX_FC/65269fe/DellEMC_VNX_FC/376 : SUCCESS in 1h 13m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"c0d5d8f0866b006dfce1a16808a704b37b0d6289","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-11 16:23:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3e71af96eccb2c1b51a2839e5e4f323d31f00946","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-11 17:54:03.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/769546-1/HPMSA-iSCSI : FAILURE in 1:04:17\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/769546-1/HPMSA-FC : SUCCESS in 37:10.36","accounts_in_message":[],"_revision_number":1},{"id":"94cf0112ba78f34a7339062ddc745793bbd9d737","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-11 19:20:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_XtremIO_FC/1a70f68/DellEMC_XtremIO_FC/731 : SUCCESS in 2h 04m 22s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_XtremIO_iSCSI/aa67288/DellEMC_XtremIO_iSCSI/731 : SUCCESS in 1h 40m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"419611d26a92eddcc3579afcc602afd2ca458b21","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-11 19:26:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c35c321d26dd8abf13026ba732e524bc90d929f3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-14 04:52:15.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/46/769546/1/check/DellEMC_PowerMAX_FC/8153dc4/DellEMC_PowerMAX_FC/450 : SUCCESS in 4h 40m 21s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/46/769546/1/check/DellEMC_PowerMAX_iSCSI/4f0e8bb/DellEMC_PowerMAX_iSCSI/414 : SUCCESS in 5h 18m 46s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"b058b5e5f9cfca41f56a708293d954df3187aac1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-14 06:08:59.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/46/769546/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"723f1a242aad2d099b8a2f7e5cedee7ecdc084a4","revisions":{"723f1a242aad2d099b8a2f7e5cedee7ecdc084a4":{"kind":"REWORK","_number":1,"created":"2021-01-06 14:38:47.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/46/769546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/769546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/769546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/769546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/769546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/769546/1"}}},"commit":{"parents":[{"commit":"9c1b4f40ba604904d664f6e60fbb6339f89c9a37","subject":"Merge \"Add docs and update support matrix for Hitachi driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c1b4f40ba604904d664f6e60fbb6339f89c9a37"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-01-06 14:13:36.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-01-06 14:13:36.000000000","tz":-300},"subject":"Remove __unicode__() from CinderException","message":"Remove __unicode__() from CinderException\n\nThis is no longer needed in Python 3.\n\nChange-Id: Ie04d6dbdb9d18047d9167f85b4346d15babca4f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/723f1a242aad2d099b8a2f7e5cedee7ecdc084a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/723f1a242aad2d099b8a2f7e5cedee7ecdc084a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
