)]}'
{"id":"openstack%2Fcinder~773694","triplet_id":"openstack%2Fcinder~master~Icf9f6409009250f766f619019252617fc9b3e1e7","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"Icf9f6409009250f766f619019252617fc9b3e1e7","subject":"RBD: Pass bytes type for mon_command inbuf","status":"MERGED","created":"2021-02-02 14:35:48.000000000","updated":"2021-02-18 16:16:33.000000000","submitted":"2021-02-09 20:21:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773694","meta_rev_id":"a9b4f399433218685b074cf9bd271528716d0e17","_number":773694,"virtual_id_number":773694,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 20:21:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"recommended":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":1,"date":"2021-02-18 16:16:33.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2021-02-08 14:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-08 14:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":2,"date":"2021-02-09 07:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":1,"date":"2021-02-09 07:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 14:46:47.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-02 15:39:37.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 15:50:54.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 16:00:29.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-02-02 16:42:34.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-02 16:47:47.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 17:50:47.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 18:38:21.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-02-02 19:18:39.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 21:36:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 21:45:53.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-02-02 21:57:31.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-02-02 22:25:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 22:35:24.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-02-02 22:54:24.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-03 01:49:22.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-03 14:06:34.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-02-03 15:43:12.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-02-03 16:23:36.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-02-03 18:49:05.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-06 18:39:58.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-02-06 23:17:28.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-06 23:52:44.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-08 14:22:42.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-02-08 14:22:42.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-08 21:22:15.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-02-09 16:10:58.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-02-18 16:16:33.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"7ce90e8257e9c4cf5524a00e20a9741b7fb3eb4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-02 14:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c67f603be39e60e8be78e3987c95c7cc8b5718af","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-02 14:36:16.000000000","message":"Patch Set 1: Workflow-1\n\nEnsure this passes on Ceph Pacific CI jobs and older Ceph CI jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46e0cecac19fb9e212d2afdb717a6aa68fe49122","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-02 14:46:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/94/773694/1/check/DellEMC_Unity_FC/0dcceb7/DellEMC_Unity_FC/571 : FAILURE in 10m 37s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/94/773694/1/check/DellEMC_Unity_iSCSI/28e5e1d/DellEMC_Unity_iSCSI/534 : FAILURE in 8m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5815baaeb2176639c7bd0ebcdc0115ac6fd5e56f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-02 15:39:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/94/773694/1/check/iscsi-driver-primera/4057cf2/4057cf2.html : FAILURE in 1h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5f1b5390ef54a7ee38b8ee3766c4e30fdbae1fd3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-02 15:45:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/94/773694/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/87173c6/87173c6.html : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d24dbece8966d30fa395f1659b3abdaaeff74b73","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-02 15:50:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/94/773694/1/check/dsvm-tempest-zadara-driver/177cdd9 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2c099f2133f969df0e0ae6c8c71b2ecb57db2a5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-02 16:00:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/94/773694/1/check/dsvm-tempest-cisco-zonemanager-job/558b448/ : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":1},{"id":"614c451e1abb1da837fb28f19327288c210a1c10","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-02 16:42:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66:9090/94/773694/1/check/dsvm-tempest-TOYOU-ACS5000-driver/9fd4205 : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b7011c9327e26f97fdd7a0ad3d994f4e58efb2ed","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-02 16:47:47.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/773694/773694/1/check-not-vote/ext-cinder-zuul/5bc2bc6 : SUCCESS in 40m 27s","accounts_in_message":[],"_revision_number":1},{"id":"63ec8c25d3fe2f2b8a1ed924895192191e11014a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-02 17:50:47.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-94-773694-1/ns5_nfs : SUCCESS in 3h 14m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-94-773694-1/ns5_iscsi : SUCCESS in 3h 7m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c666fdf002dd4c3f2f555f009165891c311075d5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-02 19:18:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/94/773694/1/check/DellEMC_XtremIO_FC/2557a9c/DellEMC_XtremIO_FC/640 : SUCCESS in 2h 05m 00s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/94/773694/1/check/DellEMC_XtremIO_iSCSI/3a1efbe/DellEMC_XtremIO_iSCSI/636 : SUCCESS in 1h 38m 45s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"253a79c4d35d5b344c97f87ae44c7943f4584427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:36:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f39b13211d254e00ae637db8ed5a1b90 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86269cd1e34a4a1e86e39779ee1369f4 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/490f52f8c9dc42d08454dce2995ddb8a : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35fa82cd57d4481589599e382a56eb97 : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fb7805abee34854a8de4b82476439f6 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7b74c833854d3e97bd486ea77b2254 : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/0b6383834572465c8587e330d0ef051d : SUCCESS in 55m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f30f85495f754dc99e959daf202f1229 : SUCCESS in 1h 10m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/739a416be19e4312a8a81eed3054cba2 : SUCCESS in 12m 17s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b499ac33277b4fcfac1ef44a15026606 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f9bdf7c4a124af3b8baba03e496ade2 : SUCCESS in 6m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b145c2ebfda749d289b07fc59f256988 : SUCCESS in 6m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6d6d0dea16234f37bfdd3004aca08d22 : SUCCESS in 38m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22d471015f0f4bffa9564aa2a7e12bee : FAILURE in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e02821b1f174e4baa638b8f339b7310 : SUCCESS in 1h 15m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5424b45f295649638a75d4be5886ed63 : FAILURE in 1h 17m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b849d4ad6b384c988e91afe7161654b4 : SUCCESS in 1h 32m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/550338608a5e4839a871a352ff331725 : SUCCESS in 1h 08m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7630d80597f24b3f99ff72559bc7e3e8 : SUCCESS in 1h 29m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a0ba232942b43738b5d7ea3df64e702 : SUCCESS in 1h 16m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d6c79931e64b4a12b5feb8cdf7d50e18 : SUCCESS in 2h 41m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d04284f3a6194cc3bba3c4f8cfa781f6 : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4d04c354fc9504bd5d61560e6d2843588c3f981e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-02 21:45:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/94/773694/1/check/DellEMC_SC_FC/3b0019a/DellEMC_SC_FC/268 : FAILURE in 8m 42s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/94/773694/1/check/DellEMC_SC_iSCSI/e26111f/DellEMC_SC_iSCSI/268 : FAILURE in 7m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"92408f72de51e0db4fdd8c9f94d86df0a5672bca","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-02 21:48:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/94/773694/1/check/DellEMC_Unity_FC/2f961aa/DellEMC_Unity_FC/575 : FAILURE in 10m 55s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/94/773694/1/check/DellEMC_Unity_iSCSI/ad1afaa/DellEMC_Unity_iSCSI/538 : FAILURE in 9m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a3b22e9621cb758fb429f974a433c6aa7d6c8923","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-02 21:57:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-94-773694-1/1612301808 : SUCCESS in 20m 42s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"3ef0ede58b62527b3c1429d373a3f1d60942121f","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-02 22:25:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/94/773694/1/20210203051120/NEC-FC : SUCCESS in 2h 08m 51s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/94/773694/1/20210203051120/NEC-ISCSI : SUCCESS in 2h 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0c458f9406017bb27657254318deda6cb2600876","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-02 22:35:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/94/773694/1/upstream-solidfire-check/cinder-solidfire-iSCSI/bf590e1/ : SUCCESS in 57m 08s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"01df5e42c635727654fe978b322991da8252d7af","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-02 22:54:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/94/773694/1/upstream-check/cinder-cDOT-iSCSI/b4159ef/ : SUCCESS in 41m 19s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/94/773694/1/upstream-check/cinder-cDOT-iSCSI-IPv6/7620859/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/94/773694/1/upstream-check/cinder-cDOT-NFS/c11e129/ : SUCCESS in 42m 19s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/94/773694/1/upstream-check/cinder-cDOT-NFS-IPv6/c13c773/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/94/773694/1/upstream-check/cinder-cDOT-FCP/bcb9fa5/ : SUCCESS in 1h 16m 28s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fdc4991fe54bd1d9359155c2aebfc2ea6d9c567f","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-03 01:49:22.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/773694-1/PVME-iSCSI : SUCCESS in 52:45.06\n* PVME-FC http://os-logs.tristero.net/dell-ci/773694-1/PVME-FC : SUCCESS in 47:26.47","accounts_in_message":[],"_revision_number":1},{"id":"4eec7b6e288775ce9e7f075908b2afa648a2aee9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-03 04:57:45.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/94/773694/1/check/DellEMC_XtremIO_FC/cc4cc74/DellEMC_XtremIO_FC/644 : SUCCESS in 2h 05m 43s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/94/773694/1/check/DellEMC_XtremIO_iSCSI/cd532b1/DellEMC_XtremIO_iSCSI/640 : SUCCESS in 1h 39m 30s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"af53a6501125658d2d41934ab5d066fa779e3c8a","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-03 13:35:09.000000000","message":"Removed Workflow-1 by Eric Harney \u003ceharney@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"a8746768d13377d022c140b1d4499a6d1fca5bc7","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-02-03 14:06:34.000000000","message":"Patch Set 1: Code-Review-1\n\nCode looks good, please add a release note","accounts_in_message":[],"_revision_number":1},{"id":"2240823b1b9475612a3143c941c15f89c878d412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-03 15:06:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4f5f2877f71d82c4168da05ec10d17ef51b493d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-03 15:43:12.000000000","message":"Patch Set 2:\n\nSo does it mean the code is backward compatible because that argument was always meant to be bytes instead of str?","accounts_in_message":[],"_revision_number":2},{"id":"a7fd9629632f219fb3fdb304dd58416928cdc610","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-03 15:49:52.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/773694/773694/2/check-not-vote/ext-cinder-zuul/f9b412b : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":2},{"id":"c4757963b28876d1bb8bd0fc474f5830e9a4e3c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-03 16:10:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/94/773694/2/check/iscsi-driver-primera/9671ef8/9671ef8.html : FAILURE in 1h 02m 12s","accounts_in_message":[],"_revision_number":2},{"id":"bdcac00d8044a2fdf43e38a6f216f6b31604b51c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-03 16:15:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/94/773694/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/560726e/560726e.html : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0c1785dd40352c09194fc58a114a97d3cd81b462","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-03 16:23:36.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/773694/2/iscsi/ : SUCCESS in 1h 07m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/773694/2/smb/ : FAILURE in 1h 15m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6a8a7b0d4c29ea5f71a8ae41f573dea146d0c418","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-03 16:28:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/94/773694/2/check/dsvm-tempest-zadara-driver/60fde86 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"123e38d86b460357a34a2c1eee950d57380b8af1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-03 17:14:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/94/773694/2/check/dsvm-tempest-cisco-zonemanager-job/ced7178/ : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":2},{"id":"374adbefe99081c3d56244e325a35d7f3473c537","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-03 18:21:10.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-94-773694-2/ns5_nfs : SUCCESS in 3h 9m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-94-773694-2/ns5_iscsi : SUCCESS in 3h 1m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"ded7a7231794d6c98d360144f3e7537218072cbb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-03 18:49:05.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/94/773694/2/check/tempest-dsvm-full-focal-py3/3a131b7/ : FAILURE in 1h 42m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/94/773694/2/check/devstack-plugin-ceph-dsvm-focal-py3/a06ebe7/ : FAILURE in 1h 33m 41s","accounts_in_message":[],"_revision_number":2},{"id":"37f95e9e95723b5ae2f6cb9b86b1e9e6c9ad37c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 21:32:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85ba2ecbe138437c9df0e7e8c794fde9 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a618ac56104a437ba72d30071bc6e168 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91faa0fd6ef1484a841944f94af6aac4 : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a7b190d5c594c94a4ecb7f9bb4712db : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b129995d5a5e47a6982dbd937052b373 : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597e3a64d32745ac94601a02bfa5a118 : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/cb9f7d4700d84a68b66cd28d2d96aa1d : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85e3c0d3dcc04a87b94aa8a0d4202084 : FAILURE in 23m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0632caab140d4216a18aa86014723168 : SUCCESS in 5m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a1ae5d9fcc154b7fa73c57ecada4dd60 : SUCCESS in 9m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1106d9b0fbda4663b538346ed0896d5a : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fece37255f7b4c358f8d9d6a58628fd9 : SUCCESS in 6m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9de39525f97b4e2bbca289450bc1fac4 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1dfb424fce9a4ab7b3fa4b830e72c141 : SUCCESS in 44m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1f609c86eeb24614a3776896d88648ce : FAILURE in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/36685c3fe528410d9544fe70b13ce695 : SUCCESS in 1h 04m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c04da89ef8f24d60847816357a172915 : FAILURE in 1h 35m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cec1b8eab1bb45f1bec845dacb3ec133 : FAILURE in 28m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7ca77eb9d6f144f69485b73cc1982982 : SUCCESS in 1h 15m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3d47957aa16a4acaa14c3f829a4af07f : FAILURE in 25m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6360276e0f8b4046b0b599e4ccaa2618 : SUCCESS in 1h 18m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/832a048410f04f6d8cebb628c13b9132 : FAILURE in 23m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cde3981f57f249b989fa7fba684a254e : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8bf76df8846d8f479806dde57930b8148e2794a4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-04 01:20:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_XtremIO_FC/a88dbc8/DellEMC_XtremIO_FC/655 : SUCCESS in 2h 04m 48s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_XtremIO_iSCSI/ba29820/DellEMC_XtremIO_iSCSI/652 : FAILURE in 19m 15s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"01c8fc38c274c93504e11a73f663daca6e6e71e5","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-04 07:06:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/94/773694/2/20210204134003/NEC-ISCSI : SUCCESS in 2h 11m 21s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/94/773694/2/20210204134003/NEC-FC : SUCCESS in 2h 26m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2c0fa4fa4294c8c0b032851a221d9201fc45ec6f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-04 15:43:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"59cd51d53ea819d978896b1f3ee1f5fb981f6b36","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-04 15:44:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1329cc427868e31c4aba67b90e8dc0ee52fc6ca1","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-04 16:49:15.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/773694/2/iscsi/ : SUCCESS in 1h 04m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/773694/2/smb/ : FAILURE in 1h 03m 17s","accounts_in_message":[],"_revision_number":2},{"id":"32ec851d459a1d4733a1814c884be99ac6963788","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-04 17:07:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/94/773694/2/check/dsvm-tempest-cisco-zonemanager-job/3597447/ : SUCCESS in 1h 19m 02s","accounts_in_message":[],"_revision_number":2},{"id":"997ada0aa301b7fa247d5fc0cf2e94e5f7563c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 22:16:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/577f59cf9dd2407880c6be5c8979afa2 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d011da7b4a604d5abf11a2c86920f48e : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8071e07081244c3b270cb9f4af0b479 : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79390badfea84422a3cf439a9b673860 : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90ba1dcbd468451cabd984ec0eca4a65 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c698acc5d54edf926b69b3855ac20e : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/9fe1a83c477d47589bc95c8057282cd2 : SUCCESS in 49m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ecfb979f98094605b4cdeb930fff8e9c : FAILURE in 1h 19m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccc75d93d4204c2b9be47258fd9300b4 : SUCCESS in 6m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/de0bda0afeaf45d5bcdc36a7c3f1b929 : SUCCESS in 12m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0ff2b91c04944654bc524df8ffd0b717 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6104b0aa85543d5b23e06f830b6b4a8 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e0424378b864ab5aac6c2e4a1a6de51 : SUCCESS in 5m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e21caf12f3f3450384fb483b95e89086 : SUCCESS in 44m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8983515658b4cb2b65c7db436ef889c : FAILURE in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6ebe2f1e58674fed854a4a3f587c2481 : SUCCESS in 1h 08m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/275176fe78e3419d8a95337bd377d1b2 : FAILURE in 1h 22m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/11a699c06b6441ffa33ffe8bac04934d : SUCCESS in 1h 31m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a2d4bd527e0249d3ac9deb7ae2743ddc : SUCCESS in 1h 11m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/68214b7412214336b75c1546e9894979 : FAILURE in 1h 27m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a07f4915713c492b809ebfdfc00fd31d : SUCCESS in 1h 18m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26f755e7697a4310876e9ea25df6cfcf : SUCCESS in 2h 15m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cdfeb45df294ab8abdd1a82f9113452 : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":2},{"id":"9b066612fc9dbf0f5dabd0ebedee77e1f615d893","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-04 22:19:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"227ad6860a1a03dd47a7cb902080af0df789c492","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-04 22:20:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31923302d6cbedeca7ac0ff2d32accd7472cf65d","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-04 23:22:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/773694/2/iscsi/ : SUCCESS in 1h 02m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/773694/2/smb/ : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e88c5fecbb43a183c7d938d5b3645655a9f2205a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-04 23:43:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/94/773694/2/check/dsvm-tempest-cisco-zonemanager-job/c76b769/ : SUCCESS in 1h 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"24da2f9df09f59492d796cef726455db29361670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 01:49:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a48908ea9ce046258396cc0bbe42f49e : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa0824f85c6844aa8a4d6f19735fb61e : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a33137e30c34ce4a37a491131d2673f : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a711bf90ee6487eb2af8017ed537a48 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e6a17fb8cad444ba53ac6e588b6e4e5 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2095bbf053904f65850fc68db8203b4b : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6daefa8e305e4c1ab723ef237d1f0b64 : SUCCESS in 57m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78a8827131bd42399beb210e66fa2a93 : SUCCESS in 1h 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06155123ac4a4c21942203e1c231d6c8 : SUCCESS in 6m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/802b1ae0d8c14f8e93d6a97cc828a3c2 : SUCCESS in 12m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79c2a806f598415f89eeeaf656ff7b35 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b78c5f4b9eb4403683ae81c0d9932728 : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0694cd4bf64402e86f7a35dfc72f879 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02ab248d7cdd4ee5a205b59af07b371f : SUCCESS in 39m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4725a448c3584e6195a87c11a29bc27d : FAILURE in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c011ec219bf4100858dbf155d2fc23d : POST_FAILURE in 1h 04m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a5a35cbbdd364b06a83e88db09192524 : FAILURE in 1h 29m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2aa6e47a40e4f58a6fc8a5f928919c0 : FAILURE in 1h 42m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/66f40efdf32f4acb973fe8d99ce82263 : SUCCESS in 1h 12m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8af0f2bee384d3cbe0afd43b39a3c77 : SUCCESS in 1h 28m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db7d52890a6e4f5c8dfcffc729c9f10f : SUCCESS in 1h 16m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92f7e4a2fd424dd09f546b6e496708f3 : SUCCESS in 2h 10m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d447604c98e94540af33615150edade5 : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":2},{"id":"bde831a150100cb3fa146aeefaae7cbd7991fdbf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-05 08:48:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8c9fd7608f7674d0f266ee46fcdac5b8c80e66bd","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-05 08:49:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e013c8aebac5336be606b6f6a5fa85117bbd45ce","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-05 09:03:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/773694/2/iscsi/ : FAILURE in 13m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/773694/2/smb/ : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":2},{"id":"59b8ca25ef1f882a51a6c4645bc2806b2540dd6d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-05 10:12:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/94/773694/2/check/dsvm-tempest-cisco-zonemanager-job/fee4ea1/ : SUCCESS in 1h 18m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1812cb1f5412fb92994e947a51e20e661f5dcd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:58:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d4359f33c584c68a3e52f560e977bec : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12adeadcb3e4973a6d9d43136c325bd : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/538b2ad04a1d461dbdf57a0d4909bedc : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98e1103895f6444fbf9f9faf2d42ee12 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1997c39772174fb2aac7a157b798c6cd : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73552b22126743f1be2aae23284fe38a : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/41004c87abff48dfbcfd547c85907e00 : SUCCESS in 1h 03m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78fe4bce75924a638987589b61159c24 : SUCCESS in 1h 21m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac2980d063dd4643917a0a30193c27cb : SUCCESS in 5m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e8b49a335c0f424aba77d3d31852b056 : SUCCESS in 11m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d27e3fb086394a7b978babca0ad076fa : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8c40ac4c0004cca84a39971602daa81 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ac1c3dd1eba47bd869606dfc9f3398a : SUCCESS in 7m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f62ab3f1457b4abda9dd97ae90a24cc4 : SUCCESS in 40m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1550c87fbb404519ad3b6eaf7a5c8ba1 : FAILURE in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70a18ce714a44baaa56b259509a838d6 : SUCCESS in 54m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/df58eb59299640f49ea7179abb090b42 : FAILURE in 1h 24m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b33cee901b5949a794b068758e04eb7f : FAILURE in 1h 17m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd8154b7a88e46bb829b5e70afa80fd7 : SUCCESS in 1h 06m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/654f3c9700bc40a0b84c92ba141b4b56 : SUCCESS in 1h 26m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a971b426644485ba46b2fb92974f84b : SUCCESS in 1h 26m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc5de2b493b14d809aeac915dfea3597 : FAILURE in 2h 07m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e5673c2e36f456faa9c58c7180a2581 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b89c565dc94f5156266032d15bc3a663e8522257","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-06 12:17:11.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/773694/773694/2/check-not-vote/ext-cinder-zuul/24aadba : FAILURE in 1m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d1e2cfcefc1f34452494bcb085d3bdd178b2c4d9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-06 15:36:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7620ac4b15f302c14c2c9403319ae84b16cc6772","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-06 15:36:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"685a47a9740e08ba1f9eb77ecba9322a93c84138","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-06 16:56:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/773694/2/iscsi/ : SUCCESS in 1h 20m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/773694/2/smb/ : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ba04b1f6d8c7b3d632c00c722552c2067e6d8451","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-06 17:01:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/94/773694/2/check/dsvm-tempest-cisco-zonemanager-job/67faef6/ : SUCCESS in 1h 19m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c2ca0ca4ab1afe1bb42fb0a60ccd4039db2ab5c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 17:44:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e070ad1cc97434e83d3353c4d5f8ff6 : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/147d0e6b6f3b4c4185cad0953ab7efd8 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3572a1446488426db78395a8ddf96e7a : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24b1b3f850184b9b80a9505fd03b685b : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/359fb8e1b0e84100ae167a72a471b2a7 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6b43d088bf45799d23f024b225556b : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/0171a55286894f7fbb4d1a9549bb1862 : SUCCESS in 1h 01m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/143258c7ac234d77869afe0b90d88293 : SUCCESS in 1h 01m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50ddbd1feb14c8d804d2544c18fca6e : SUCCESS in 5m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/57365bcce9b24704a5386b8f4534ba9b : SUCCESS in 12m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8e7f975e74842ffb9c902335e11827d : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3982708cfaa14c2a88b00b0db99bb033 : SUCCESS in 6m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3481c5da18bd415f8e1690e9ad4cc45e : SUCCESS in 8m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/687a865151e94c8c8d61fefa8e5a9bc4 : SUCCESS in 42m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e46635ee69344d13a68985eb89448449 : FAILURE in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5bdbcea728974f4d96d5ca9a30ff12bb : SUCCESS in 1h 12m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3f3f866a9e9d46748b00328ae3fe37ca : FAILURE in 1h 28m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/07be452558c24af8a724673999ec22e3 : SUCCESS in 1h 43m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8548e07db4f641ad87809c0ca7a0b317 : SUCCESS in 59m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b1fad1febe414d4fa4c559cb07a3f361 : SUCCESS in 1h 12m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/11b9811a4e8440728427026161f1f924 : SUCCESS in 1h 15m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ef3f38f7af14720b58ce6c5870e9644 : SUCCESS in 2h 06m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85334489b629465dbae01f987b9432ce : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b8b72eb8d2b780d99b28a5b55a10ee304599501f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-06 17:50:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/94/773694/2/upstream-solidfire-check/cinder-solidfire-iSCSI/1b0cec4/ : FAILURE in 4m 45s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"a58ddf65ddf69a6f7a65eddd2cc10b068ae1ed18","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-06 18:39:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_PowerStore_iSCSI/3d7bf75/DellEMC_PowerStore_iSCSI/355 : SUCCESS in 54m 54s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_PowerStore_FC/a1ee4d1/DellEMC_PowerStore_FC/349 : SUCCESS in 46m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"f5063d0ec94786e70adf47c1627f67589caf7af0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-06 18:53:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-iSCSI/c10b753/ : SUCCESS in 36m 28s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-iSCSI-IPv6/f81a463/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-NFS/c5cda4c/ : FAILURE in 1m 38s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-NFS-IPv6/af88a12/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-FCP/fa3776f/ : SUCCESS in 1h 07m 26s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2b36e3ff0e01d689ddf9dded8a993272c54edfce","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-06 21:28:28.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/773694-2/PVME-iSCSI : SUCCESS in 50:33.97\n* PVME-FC http://os-logs.tristero.net/dell-ci/773694-2/PVME-FC : SUCCESS in 49:13.48","accounts_in_message":[],"_revision_number":2},{"id":"5b355344b05c02a8a67ea6dc78655691fde187d2","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-06 21:33:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_SC_FC/24fb81d/DellEMC_SC_FC/295 : FAILURE in 1h 53m 42s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_SC_iSCSI/c6099bd/DellEMC_SC_iSCSI/295 : FAILURE in 2h 15m 12s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"ef278067c1dcca8d7fa5423a26c615b1e1a746ee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-06 23:17:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/94/773694/2/check/DellEMC_PowerFlex/646e716/DellEMC_PowerFlex/341 : FAILURE in 2h 44m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"f7ac597baaf332b6c0d1ab2dd020a7ae64ad742e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-06 23:26:46.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_Unity_FC/f872aa0/DellEMC_Unity_FC/613 : FAILURE in 1h 37m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_Unity_iSCSI/9150ccf/DellEMC_Unity_iSCSI/578 : SUCCESS in 1h 36m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"41f942a4b05b40813e5d2c7ebb7738f684cb959f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-06 23:52:44.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_VNX_iSCSI/c258eee/DellEMC_VNX_iSCSI/341 : SUCCESS in 3h 07m 13s\n- DellEMC_VNX_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_VNX_FC/762d114/DellEMC_VNX_FC/346 : SUCCESS in 1h 11m 46s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"2e61c370b827d8439d7022a825b88b770226100c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-07 05:03:33.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_XtremIO_FC/92d1665/DellEMC_XtremIO_FC/690 : SUCCESS in 2h 03m 26s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_XtremIO_iSCSI/31d2b84/DellEMC_XtremIO_iSCSI/689 : SUCCESS in 1h 39m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"3f7dcf35c4367ef26cfcd0016970e1dd0e0df224","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-07 15:25:44.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/773694/773694/2/check-not-vote/ext-cinder-zuul/f853109 : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f2709ebb7cda690234516d4960f80ba326d87d60","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-02-08 14:11:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1860687d41cef2d460ab6f3c5ad78cbd27e3c42b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-08 14:22:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"19df46a2bee9501289d9aa7f00238567203cf277","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-09 07:01:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM.","accounts_in_message":[],"_revision_number":2},{"id":"7cfb3e5fcb5a1bf858ad41d5e8b03f4446885942","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 07:02:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"91e6792c31d8c503ee4e6be610ed505e79d4bd63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 08:13:05.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e755aba3d624a7eb7b30142239cc925 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ee3519e3044a208ca045b062428ed0 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff1cb960eb4f4c05a308a46809272f73 : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e305e07d51c04ffab88b4b1a8143823b : FAILURE in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19107f2c9d19430689cfbde2266902d6 : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/045bdf7c437f4090877a27a0cdc6b267 : SUCCESS in 56m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ce0bc85979e420a84c3ef52695cb94d : SUCCESS in 1h 07m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81b6fab5215e4a2d9458694b08219fac : SUCCESS in 6m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c873cda6e75b4fbab48c9cf71b2c6050 : SUCCESS in 1h 07m 42s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b94d5417ed1434ca0ebc59583c9af0d : SUCCESS in 1h 05m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f91391d38ef247bab4dcf755eeb70b54 : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":2},{"id":"05a0de4a4214590afa62631364edaa0cb60ff9d1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-09 10:22:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68763b798bc6068964d4c405cbf72dfe86f07549","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-09 10:26:42.000000000","message":"Patch Set 2:\n\nThe gate failure is due to a test expecting LOG.info to not have been called but it can be called by various other methods in the duration of this test run, rechecking should work but the test should be fixed as well.\n\ntest_log_create_vol_from_snap_without_v2_clone_api\nAssertionError: Expected \u0027info\u0027 to not have been called. Called 1 times.\nCalls: [call(\"Purging trash for backend \u0027%s\u0027\", \u0027RBDDriver\u0027)].","accounts_in_message":[],"_revision_number":2},{"id":"43bf69e6281c76afc41e55bef30ba8c72477a196","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-09 10:28:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d06685dfd3228cedf08fdfd6aeb77977ecb9380","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-09 10:34:17.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/773694/773694/2/check-not-vote/ext-cinder-zuul/be1ca6c : FAILURE in 1m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d04d7b846a084b1b9d21adeecef723b21eb2c715","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-09 10:54:26.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/94/773694/2/check/tempest-dsvm-full-focal-py3/87a6a47/ : FAILURE in 25m 43s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/94/773694/2/check/devstack-plugin-ceph-dsvm-focal-py3/7ecf5d3/ : FAILURE in 25m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7d6c8689057559698e9a293743deabd3a1721809","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-09 11:37:49.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/773694/2/iscsi/ : FAILURE in 1h 05m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/773694/2/smb/ : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4b0679b38c82c852bb687e757cdf5b54226edace","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-09 11:52:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/94/773694/2/check/dsvm-tempest-cisco-zonemanager-job/76f6d69/ : FAILURE in 1h 22m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7eb674aa17c158ae02871394b8b949fe185fb46a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 13:52:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/434308a7ccb4450399ed4ec1b6525d76 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651b31c99db344db8bd625ea397fd2ba : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e96f424e82e74070b8b688acdeab6de2 : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a39cc97d3e7d45a08439fdf0e4375c8c : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd2dbe936c6c46319af1cb43cc09a47c : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb09cf15b6224f74835067efea0e614c : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/01411b0f45654f2cb0336c3cae0742e4 : SUCCESS in 58m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0148c3335764a64b62f209788918466 : SUCCESS in 1h 17m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5effba370fe4496bc0697de4f611a8c : SUCCESS in 10m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5a99e9e1c272431595fc7135b0cb256d : SUCCESS in 15m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/071263791851426997cf3e06f97ee631 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe39ee0f79e4448b89c5a93379227808 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a96f2f150f8446449de039a39e8b6f75 : SUCCESS in 7m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0e7e7389c5c0430cb1cb5fd1db1b3984 : SUCCESS in 41m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60934a2214f64b3aacf3342b7d0c8cc4 : FAILURE in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7cc0d67bc87e461a8bde369ac06a2be4 : SUCCESS in 1h 00m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d2db219d5b94f64b7c62150c9b95676 : FAILURE in 1h 38m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cfbda56e73904b029e6fb6c14b93b1bd : SUCCESS in 1h 34m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3fed60cda98248ef94281b6210d2191f : SUCCESS in 1h 18m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/848610ef2e8e4830add869f744638b39 : SUCCESS in 1h 22m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1112f36d7b4b4ed6a9b0ecb83c2b019b : SUCCESS in 1h 19m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbf981f35db6476baa30bed4f94edae2 : SUCCESS in 2h 02m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2ad0f557aff4400964536fc52d3cae4 : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a97897ed60a9994b28d1fc271e1a965ad7d0d8cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 13:53:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a004fc4808b8e9dcefbc987360c173ff28201ef1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-09 14:06:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/94/773694/2/check/DellEMC_PowerFlex/3fa4a59/DellEMC_PowerFlex/351 : FAILURE in 13m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"695eb0d83962caa1f354f8981d45903b31900a69","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-09 14:17:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-94-773694-2/1612878729 : SUCCESS in 24m 53s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"4e43401808f78222b8810404c9aae719592ce5b2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-09 14:44:20.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_PowerStore_iSCSI/4d3fbe6/DellEMC_PowerStore_iSCSI/365 : SUCCESS in 51m 45s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_PowerStore_FC/d6e1563/DellEMC_PowerStore_FC/359 : SUCCESS in 45m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"8ea6e082a327f8dafef491e2b947189a8eda5790","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-09 14:58:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-iSCSI/af3a0d0/ : SUCCESS in 37m 29s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-iSCSI-IPv6/49d806d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-NFS/5369e41/ : SUCCESS in 37m 06s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-NFS-IPv6/fd9e5ea/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/94/773694/2/upstream-check/cinder-cDOT-FCP/c5d28d3/ : SUCCESS in 1h 04m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b7e088d0628dd761e62ca2fc09fa37dcdbe21b73","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-09 15:15:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/94/773694/2/upstream-solidfire-check/cinder-solidfire-iSCSI/e2f776e/ : SUCCESS in 1h 21m 34s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"f80b3dc24444649e2c95ba84d3702fa33cd4e6fa","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-09 16:01:39.000000000","message":"Patch Set 2:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/773694-2/PVME-iSCSI : SUCCESS in 49:25.50\n* PVME-FC http://os-logs.tristero.net/dell-ci/773694-2/PVME-FC : SUCCESS in 48:33.84","accounts_in_message":[],"_revision_number":2},{"id":"e5317fece868ca15594c480c4447020673447bc7","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-09 16:10:58.000000000","message":"Patch Set 2:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/773694-2/HPMSA-iSCSI : FAILURE in 1:06:35\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/773694-2/HPMSA-FC : SUCCESS in 39:27.85","accounts_in_message":[],"_revision_number":2},{"id":"cc13108c778be726273077d62a63a9dbbe04f1ff","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-09 17:00:10.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_Unity_FC/991c54d/DellEMC_Unity_FC/644 : SUCCESS in 1h 23m 49s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_Unity_iSCSI/44cf550/DellEMC_Unity_iSCSI/606 : SUCCESS in 1h 39m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"2b843a2daf24ca0c4c71a6fc3b8536d5e322cdd3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-09 17:03:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_VNX_iSCSI/9c82450/DellEMC_VNX_iSCSI/350 : SUCCESS in 3h 10m 49s\n- DellEMC_VNX_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_VNX_FC/866360f/DellEMC_VNX_FC/355 : FAILURE in 15m 40s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"410e59ae6e18943dcc7f3785f306db81cf2adcc3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-09 18:41:17.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/94/773694/2/check/DellEMC_XtremIO_FC/1199515/DellEMC_XtremIO_FC/708 : SUCCESS in 2h 05m 26s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/94/773694/2/check/DellEMC_XtremIO_iSCSI/b8e48ea/DellEMC_XtremIO_iSCSI/708 : SUCCESS in 1h 38m 57s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"1aea41de0eda3d724c0c55d11c80c671b56311af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 20:21:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c88444e6efd4fb7af7e89c856cf622c : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/967bcda5c22c4bd6b0ed273efb9e31b5 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fdd938019c4495f876e252babaaa8a2 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/295aa6e9452b4101b0ed3d00dc6fad20 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eabb12be993a4454b96bd06bef8e4f1c : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f58fc1b72fbc48a0ad5a4e1341efb47a : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4591ca6db60b43639443de561d97a83e : SUCCESS in 1h 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75c06141292f43c6ad7b8052bb87acb6 : SUCCESS in 6m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/93c4aff0467f4e5194c78c9b0f770489 : SUCCESS in 1h 08m 20s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/729e39c2ac9a43af98c5d31bc74ed399 : SUCCESS in 1h 04m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8b2f79659e04dde878ace12c56395be : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3810615bc1a1b8a2001cc32e2348173ca220efaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 20:21:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"10e154829f4c48ecb4b7c039c3c47c82533a3960","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 20:26:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e900bdbbe614ab8b1c9572b32315d0c : SUCCESS in 2m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7181aefd8ebd444a9eab1f703ab3a81e : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a9b4f399433218685b074cf9bd271528716d0e17","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-02-18 16:16:33.000000000","message":"Patch Set 2: Code-Review+1\n\nYes please.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"11704d0e8c4077fe6bc95db164c40857f65d49bb","revisions":{"abea27e48955ed46915ce483fd92253cd70fb27b":{"kind":"REWORK","_number":1,"created":"2021-02-02 14:35:48.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/94/773694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/94/773694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/773694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/773694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/773694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/94/773694/1"}}},"commit":{"parents":[{"commit":"677ee75d5dd5cc4374c0dd10f7c25ba128b91003","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/677ee75d5dd5cc4374c0dd10f7c25ba128b91003"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-02-02 14:23:25.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-02-02 14:35:28.000000000","tz":-300},"subject":"RBD: Pass bytes type for mon_command inbuf","message":"RBD: Pass bytes type for mon_command inbuf\n\nCeph Pacific enforces that bytes instead of str\nare used here.\n\nCloses-Bug: #1913449\nChange-Id: Icf9f6409009250f766f619019252617fc9b3e1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/abea27e48955ed46915ce483fd92253cd70fb27b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/abea27e48955ed46915ce483fd92253cd70fb27b"}]},"branch":"refs/heads/master"},"11704d0e8c4077fe6bc95db164c40857f65d49bb":{"kind":"REWORK","_number":2,"created":"2021-02-03 15:06:12.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/94/773694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/94/773694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/773694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/773694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/773694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/94/773694/2"}}},"commit":{"parents":[{"commit":"677ee75d5dd5cc4374c0dd10f7c25ba128b91003","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/677ee75d5dd5cc4374c0dd10f7c25ba128b91003"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-02-02 14:23:25.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-02-03 15:04:44.000000000","tz":-300},"subject":"RBD: Pass bytes type for mon_command inbuf","message":"RBD: Pass bytes type for mon_command inbuf\n\nCeph Pacific enforces that bytes instead of str\nare used here.\n\nCloses-Bug: #1913449\nChange-Id: Icf9f6409009250f766f619019252617fc9b3e1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11704d0e8c4077fe6bc95db164c40857f65d49bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11704d0e8c4077fe6bc95db164c40857f65d49bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
