)]}'
{"id":"openstack%2Fcinder~774116","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~Ib52879a270a4ae4cdd3cb5fc18f2b7bdbccd8ab5","project":"openstack/cinder","branch":"stable/victoria","topic":"change-774116-2","hashtags":[],"change_id":"Ib52879a270a4ae4cdd3cb5fc18f2b7bdbccd8ab5","subject":"Log information about the Ceph v2 clone API","status":"MERGED","created":"2021-02-04 17:07:50.000000000","updated":"2021-03-03 19:10:47.000000000","submitted":"2021-02-11 18:10:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774116-change-774116-2","cherry_pick_of_change":754849,"cherry_pick_of_patch_set":26,"meta_rev_id":"06d7ce51a7347f04a8163070a8d027fea06f1b58","_number":774116,"virtual_id_number":774116,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-11 18:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-11 16:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-11 13:46:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-02-11 13:34:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-11 16:13:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 17:09:59.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-02-04 17:09:59.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-02-04 17:09:59.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-04 17:46:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 18:02:08.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 18:15:49.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-02-04 18:20:01.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 19:33:11.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-02-04 20:00:30.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-04 23:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 00:27:20.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-02-05 01:26:12.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-02-05 02:35:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-05 14:04:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-05 18:20:44.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-02-05 23:05:38.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-02-11 16:13:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"827953bae1ffb94a8497c1954864cc58f0b136f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-04 17:07:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"af28b94f2bb4735a6c2a9737ec7e0fb519ea484f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-04 17:46:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/16/774116/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/894c563/894c563.html : FAILURE in 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8610e8cab3ad069f26f89fb891295413c996cdf4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-04 17:55:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/16/774116/1/check/iscsi-driver-primera/15aec0d/15aec0d.html : FAILURE in 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"25092d02bd0280667e3c83d8a03880db485adfd0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-04 18:02:08.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-16-774116-1/ns5_nfs : FAILURE in 53m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-16-774116-1/ns5_iscsi : FAILURE in 53m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"9c2148371d062952602eb4864ebc2d1092e8ff67","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-04 18:15:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/774116/1/iscsi/ : SUCCESS in 1h 06m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/774116/1/smb/ : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b60397f75463f22a32a4220a0b123fdb06ce954d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-04 18:20:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/16/774116/1/check/dsvm-tempest-zadara-driver/d3a7c22 : ABORTED in 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d2b7f176bb53bb120e9e7a12000cb508aab3b91","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-04 19:33:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/16/774116/1/check/dsvm-tempest-cisco-zonemanager-job/84b5489/ : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0f485407c80438868f9d2985f6ad5da13f19c955","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-04 20:00:30.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66:9090/16/774116/1/check/dsvm-tempest-TOYOU-ACS5000-driver/246c38e : FAILURE in 28m 35s","accounts_in_message":[],"_revision_number":1},{"id":"27ff30d3d7ad530aeabcc8ebd27004dd9afa2231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 23:52:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60573e6427824cbd8cd7767f7ae0f554 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b46308ca659a4310867f56803adfc112 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/968e92247bf04669856793699d6dd018 : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4ba2b03c850482597343f5dff54223b : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e94216366494da9ade38427cf8d4f14 : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/450770eb90b84e588f1bb2ecf3b550fd : SUCCESS in 1h 04m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b10b92d275684b92b0d22d9e6b1e39d5 : SUCCESS in 1h 12m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5266d56bbf0a4e0b85ac4fae3a5f0809 : SUCCESS in 12m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19493774972d4540bb324fd6d696d378 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e181d2288ce543e782d0ba5256797cc5 : SUCCESS in 6m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af9f00c508e54a52978037d4a16ebe36 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1b2dec497664b7896c11062e63d406b : SUCCESS in 39m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca06ecc456454ac7b5275b232b574c22 : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ae5baf41d9d146ec8ff50c175b634fe7 : SUCCESS in 1h 12m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/63e6e0e20f5542308f1c8c8e5863ecc5 : FAILURE in 1h 23m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec56acf5d0d94b6bac0104a6e4b856c7 : SUCCESS in 1h 25m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/183fd18c9b8549a58cbae51c7cf22eb0 : SUCCESS in 1h 14m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e30263736b6f4c718b1909bc5120b0fe : SUCCESS in 1h 31m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6de1d52bd55141c5b86c26252d57c0ba : SUCCESS in 1h 25m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cdc02444f6524600957c4b01691c6e02 : SUCCESS in 2h 14m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/941607af62e44937b745224ef43ca6ee : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":1},{"id":"97c5074410a278a7bb0a77d70cb3f061bc7bfa9d","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-05 00:27:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-16-774116-1/1612482777 : SUCCESS in 34m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"6b1dbf0a649b32c3c948c595991ac97774d8c16f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-05 01:26:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/16/774116/1/upstream-solidfire-check/cinder-solidfire-iSCSI/c4c67ec/ : SUCCESS in 1h 31m 55s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"8396edd3b43e181179b523fa4156554b466cb6e7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-05 01:42:50.000000000","message":"Patch Set 1: Code-Review+2\n\nI think this is worth backporting.  It\u0027s low risk, isolated to one driver, and the log message it adds won\u0027t overflow the logs and provides some important information to operators so they can improve performance of a ceph backend.","accounts_in_message":[],"_revision_number":1},{"id":"be517b2e91fc68a2ca4137d541e160a75d7490f9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-05 02:35:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/16/774116/1/upstream-check/cinder-cDOT-iSCSI/2877959/ : SUCCESS in 37m 13s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/16/774116/1/upstream-check/cinder-cDOT-iSCSI-IPv6/fd4a74e/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/16/774116/1/upstream-check/cinder-cDOT-NFS/4b619dd/ : SUCCESS in 2h 36m 04s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/16/774116/1/upstream-check/cinder-cDOT-NFS-IPv6/cd69489/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/16/774116/1/upstream-check/cinder-cDOT-FCP/ae0f471/ : SUCCESS in 1h 08m 32s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b4262e43f035bbba95fe373b3ace5ffb6b888a69","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-05 06:45:26.000000000","message":"Patch Set 1: Code-Review+1\n\nMerged in master(wallaby), minimal change, logging info in the RBD driver for v2 clone API which will helpful for victoria as well.","accounts_in_message":[],"_revision_number":1},{"id":"4762dba95eb750de4b965236b698b2ad8cc404d2","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-05 14:04:34.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/774116/1/20210205215513/NEC-ISCSI : FAILURE in 1h 06m 34s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/774116/1/20210205215513/NEC-FC : FAILURE in 1h 09m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bac7a31a6ecb1ea804c45a326928b23e476144ea","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-05 18:20:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis is a good backport.  Thank you.","accounts_in_message":[],"_revision_number":1},{"id":"4122ff7d68df51aeb1a6697533153bd429f09950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 18:21:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26a6f7433965b3574bb9b44525b8b883a27f16cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 20:26:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b69041d580024198b175ab94725f10eb : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d15a0c67cbe464f8ef1549bcae895da : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c47f62119754c60ae8f0ef76bb60362 : FAILURE in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b015dc7af1ae4e31a0ad5194f4ddb779 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38266e6f98140878e723c5aadb7aa86 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/1596925af9d04c28b42330baf079fbd1 : SUCCESS in 51m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48f133e049ef4185aa0796b10e9f084c : SUCCESS in 1h 02m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0833d572b8cd47a3be646782de3a675e : SUCCESS in 1h 24m 49s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/97a26ad91b0a4b1f93773c9ea2c05c48 : SUCCESS in 1h 43m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b772e497cd240f2871d43c63bd3a7a3 : SUCCESS in 1h 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d3c2287dc7c76cc6cfae9a1ccdc4755a442a0433","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-05 21:47:09.000000000","message":"Patch Set 1:\n\nThe error doesn\u0027t look related to this patch","accounts_in_message":[],"_revision_number":1},{"id":"d841c1a437d20c6496823112c53bfb48c2522d95","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-05 21:48:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a2fbdd5553e82f9ad9bc8daba082a81796ebf76","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-05 21:48:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30762f2d85af37ae526718391bf0b6ea41c14783","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-05 23:05:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59d11250e4b3881f416f985bc39be2388a6fd496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-05 23:05:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cb476c6c801eccdfc3becf8e8b4e7a13f38b9375","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-05 23:06:09.000000000","message":"Patch Set 2:\n\nRebased to remove the +W for now, see above comment.","accounts_in_message":[],"_revision_number":2},{"id":"46e37874ea99896e9d31bf34387d2f1f0fc981c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-05 23:25:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/16/774116/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4c07338/4c07338.html : FAILURE in 9m 12s","accounts_in_message":[],"_revision_number":2},{"id":"27c78d1fe18ff385b8ad8e5e3aa6aa1f5422ad8c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-05 23:29:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/16/774116/2/check/iscsi-driver-primera/6ab9202/6ab9202.html : FAILURE in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"6bfe5694d253a8625ace5b2c8ee79c1cd9b4b0ea","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-06 00:17:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/774116/2/iscsi/ : SUCCESS in 1h 11m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/774116/2/smb/ : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c7e8d7b020406cbaccc54f5427f071994dd4d769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 01:14:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8beced1235e4548a6a57b206d112085 : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1df8defae98e4cb090a67c6e850f415c : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e3165b1793344bfaa3bf44b84d2c507 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41c5e6a0313a40f796b6e839f50c4715 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc65ba5344b488d8f0ccc4f05e8234a : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/680272e8459c43978478b2bc4ae39cea : SUCCESS in 1h 01m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bce615bac664581806420c86356fee2 : SUCCESS in 1h 08m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2461f401f544422dbaf82ab8de76b565 : SUCCESS in 11m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3890323f1be64bfdba81eefff8835d69 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/808b89a96a3a41e2b450b92d173459a4 : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a06c2d08ac3b421eb05477f651f68261 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc2baf075a924b6f8b0e5fd97e6e77d0 : SUCCESS in 42m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db73d1df15b941529ca6d658240c2b47 : FAILURE in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b911fac5f4f4027a200108b13abe8a0 : SUCCESS in 1h 06m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ba482e0ac25943a98fbbd46744ce21f7 : FAILURE in 1h 21m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1a4e12beba2a406a8088792730b21c39 : SUCCESS in 1h 35m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6dfdb9ca022145b59e0b5f52be3854fe : SUCCESS in 1h 01m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5acfafd7eccb4baf9b3a6b9ea806aace : SUCCESS in 1h 15m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b492875972b472ebbf21959939a7f09 : SUCCESS in 1h 15m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bdfb9c2d0ea34199b15aa4d045a0bf94 : SUCCESS in 2h 04m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcfcc4fdf16e4bacb44b90824c4a32f1 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1b16f126450536266d265d1459eae6cc325664fe","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-06 01:56:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/16/774116/2/check/dsvm-tempest-zadara-driver/7d00e2e : SUCCESS in 1h 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f93d67053fbe1932e7b9e94d2da4174564e78e64","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-06 02:02:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-16-774116-2/1612574049 : SUCCESS in 34m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"5bff28e8b8877a6aa779b0e970d57ff5783c349d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-06 02:25:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/16/774116/2/check/dsvm-tempest-cisco-zonemanager-job/2323c35/ : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":2},{"id":"409d80dbaab674afb78268b65628ff567a4fe04b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-06 02:42:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/16/774116/2/upstream-solidfire-check/cinder-solidfire-iSCSI/9210d0a/ : SUCCESS in 1h 27m 04s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"52c220601ed3c306112890c59f6d4db41288c4ed","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-06 03:28:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/16/774116/2/upstream-check/cinder-cDOT-iSCSI/e64b0f3/ : SUCCESS in 37m 09s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/16/774116/2/upstream-check/cinder-cDOT-iSCSI-IPv6/7ff9c37/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/16/774116/2/upstream-check/cinder-cDOT-NFS/afbbb59/ : SUCCESS in 38m 52s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/16/774116/2/upstream-check/cinder-cDOT-NFS-IPv6/899d704/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/16/774116/2/upstream-check/cinder-cDOT-FCP/2e20482/ : SUCCESS in 1h 15m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"34a12004ad9774b5cd76e8663d4e1242aedceec6","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-06 03:33:48.000000000","message":"Patch Set 2:\n\nMerge failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI","accounts_in_message":[],"_revision_number":2},{"id":"285feda2d75ce3bb4c9e2b21d4eb63c9e7bc3476","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-06 19:56:51.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-16-774116-2/ns5_nfs : FAILURE in 2h 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-16-774116-2/ns5_iscsi : FAILURE in 51m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"64d45417b82929d6a541d6e633468e428298ea41","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-07 01:15:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/774116/2/20210207075755/NEC-FC : SUCCESS in 2h 06m 15s\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/774116/2/20210207075755/NEC-ISCSI : SUCCESS in 2h 18m 01s","accounts_in_message":[],"_revision_number":2},{"id":"54c76424c7eb979dcbbf4ecbff651e4bb165df55","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-08 15:35:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2cb8966b308eeaaf542767204b2a30ddc3640abc","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-09 10:23:41.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nAdded a comment inline regarding a gate failure due to these tests.","accounts_in_message":[],"_revision_number":2},{"id":"f571efd1f1b7d0354f4655c68d6bdb276f06f80b","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-09 14:51:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c391c7201b8405d39fbdbefd326bde027f0cc5ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-02-10 17:50:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a5d5c7e1ffaa635d0d38d3f45279fd0886e27c64","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-10 17:51:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f7e2c78302b76f2f130ef6c6aad793ace0243d6","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-10 18:09:18.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/774116/3/iscsi/ : FAILURE in 17m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/774116/3/smb/ : FAILURE in 17m 42s","accounts_in_message":[],"_revision_number":3},{"id":"bf229d6e3dc0a6993a81e7298f3fee4125fab8aa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-10 19:13:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/16/774116/3/check/iscsi-driver-primera/35f12de/35f12de.html : FAILURE in 46m 09s","accounts_in_message":[],"_revision_number":3},{"id":"ca570a8e189e408413bef498967a700782007edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 21:42:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b98160054d84dfa9f67651f18b963b2 : SUCCESS in 15m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09fb7a6bd5a44fbadc9e166518e32e3 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd3d0893d5e34e7882a23a288a30c93a : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc4daac02abd489f9f28891d1e9672d8 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcfee69cf1bf4b2d8b2a87ee3517b917 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/7913f9b50e13419c8c05067a7f9c7b49 : SUCCESS in 1h 10m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22bc5f00830643069846828bf551bc0c : SUCCESS in 1h 06m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fa298f9691824fe2aff63f52e98769a0 : SUCCESS in 14m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/108d3b8897814679ac433e0aa0db7b9b : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3cd40d6ccdc48de8ba3550bba742f97 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eeab7e259ad94974a65a08114ea1c610 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b274c9e4df1f428fb414f84d5b96cafc : SUCCESS in 39m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eabc2adb1dbc4549a850767a193eb060 : FAILURE in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4de6ec2bc230452db0c34fa077f8436d : SUCCESS in 1h 28m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/31291cafc5f2431fb947e62bc7037938 : FAILURE in 1h 19m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f816e7600444cf2923eec65b883291f : SUCCESS in 1h 39m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f7e484108f5c476f81a762d05ce4f8df : SUCCESS in 1h 36m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/65b14d1f664b41b6b40c505b164a913c : SUCCESS in 1h 26m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4436b6324909419090c36d54e2a89e3f : SUCCESS in 1h 24m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb793bfce62c49019d9f7a5c3be884b7 : SUCCESS in 2h 07m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08542f5697b443438df187ebd2686e95 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":3},{"id":"80e325346d4c07d221e36cf5aa891282b9abcb78","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-10 22:09:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/16/774116/3/upstream-check/cinder-cDOT-iSCSI/70b2784/ : FAILURE in 18m 18s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/16/774116/3/upstream-check/cinder-cDOT-iSCSI-IPv6/3d331e6/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/16/774116/3/upstream-check/cinder-cDOT-NFS/f7f2c2a/ : FAILURE in 17m 56s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/16/774116/3/upstream-check/cinder-cDOT-NFS-IPv6/a3b666c/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/16/774116/3/upstream-check/cinder-cDOT-FCP/425ee3b/ : FAILURE in 25m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"9531648a553dff13e7b7a87da3a8dda515ffdd92","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-10 22:11:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/16/774116/3/upstream-solidfire-check/cinder-solidfire-iSCSI/d18d7c0/ : FAILURE in 27m 10s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"5b5edd64e5e0fcdb5173bce541e2aba612cf4c6d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-10 22:37:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/16/774116/3/check/dsvm-tempest-cisco-zonemanager-job/fd3b3ef/ : SUCCESS in 1h 20m 36s","accounts_in_message":[],"_revision_number":3},{"id":"238f286ef2219e32b38999bc4915e6b1e6988dcb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-11 00:53:02.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-16-774116-3/ns5_nfs : FAILURE in 1h 20m 21s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-16-774116-3/ns5_iscsi : FAILURE in 1h 20m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"263bdb42a3eaf878fe424e5d79f3872e516765c8","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-11 03:21:01.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://18.181.80.66:9090/16/774116/3/check/dsvm-tempest-TOYOU-ACS5000-driver/8d4e50f : FAILURE in 15m 32s","accounts_in_message":[],"_revision_number":3},{"id":"23468f9b5fcf35c6b6f08d59f3a6891b9b132b3d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-11 13:34:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1f6e5b3d9d1eb91e20a97a84933abfb74a098137","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-11 13:46:38.000000000","message":"Patch Set 3: Code-Review+2\n\nThe original change and the squashed commit have both merged in master.  I think this is a good backport.  It\u0027s isolated to a single driver and can assist operators in ensuring that they have optimized their Ceph backend.","accounts_in_message":[],"_revision_number":3},{"id":"407a1a4a32358f4955168d6a4528ab872b1f5f89","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-11 16:13:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"a0d341c705aaa327babd088328a58003483d8bbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 16:13:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7e24749437c5b09ce7035f2a5b12ee954d9a4e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 18:10:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c3ad79233204408ae5202951c2bdb3a : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56fea603587416e98f86eb4deb4d45a : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4185b218d16946a396fef468a22b311c : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe0baa746bbb43e4a10adaaabfda4fd7 : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9241cfbc02894f24a57a36c0df6b65ee : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/7863e94e9c4a426f97b9a0ca13d70985 : SUCCESS in 58m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/345aae4373e040c7974f6ee5659af840 : SUCCESS in 1h 09m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/72d9b55eaddb47dd869c34bff6051b13 : SUCCESS in 1h 32m 07s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/038a2189cdc24c1390818613f84cf76f : SUCCESS in 1h 17m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6118884fed543d89e99ef7190e0e3aa : SUCCESS in 1h 07m 03s","accounts_in_message":[],"_revision_number":3},{"id":"952382f27b1644c44f5430d11389a4362b4003e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 18:10:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3cfc3447b073d5a9a5e2bee78093614fa79630c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 18:13:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f74fdeff1f445ab8ae71e7c6235591 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":3},{"id":"06d7ce51a7347f04a8163070a8d027fea06f1b58","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-03 19:10:47.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-16-774116-3/ns5_nfs : SUCCESS in 3h 28m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-16-774116-3/ns5_iscsi : SUCCESS in 3h 13m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d5fade7d8c6ade6166946280c7717cff10762d1f","revisions":{"362ba29c598baf492511626dd21bf35b5dec3fa3":{"kind":"REWORK","_number":1,"created":"2021-02-04 17:07:50.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/16/774116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/774116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/774116/1"}}},"commit":{"parents":[{"commit":"839cade8240df97e4e022665d4d147ccd7446650","subject":"Merge \"NetApp SolidFire: Fix error on cluster workload rebalancing\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/839cade8240df97e4e022665d4d147ccd7446650"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-28 20:20:34.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-02-04 17:07:50.000000000","tz":0},"subject":"Log information about the Ceph v2 clone API","message":"Log information about the Ceph v2 clone API\n\nStarting with the Mimic release and the Ceph RBD clone v2 feature,\nit is no longer necessary to manage snapshot protection status.\nThe usage of clone v2 is controlled via the Ceph “rbd default\nclone format” configuration option which defaults to “auto”, but\ncan be overridden to “1” to force the legacy clone v1 behavior or\n“2” to force the new clonev2 behavior.\n\nThis must be configured in Ceph, it\u0027s not something that can be\nconfigured on the Cinder side.\n\nAs an alternative, setting the minimum client to mimic\n\"$ ceph osd set-require-min-compat-client mimic\"\n\nAbove all with Ceph RBD clone v2 support enabled, image snapshots\ncan be cloned without marking the snapshot as protected because\nRBD automatically tracks the usage of the snapshot by clones. As\na result Ceph RBD clone v2 API improves performance.\n\nThis proposed patch adds logging messages for the operator\nindicating whether Ceph has been configured to use the Ceph RBD\nv2 clone feature.\n\nIn this way the operator can consult with the Ceph administrator\nto take appropriate action.\n\nYou can determine which volumes are using the clone v2 feature\nthanks to volume.volume.op_features(). It returns a value depending\non the v2 clone api being enabled or not.\n\nCo-authored-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ib52879a270a4ae4cdd3cb5fc18f2b7bdbccd8ab5\n(cherry picked from commit efd83675780a76422436d47c4622e22a0645cc30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/362ba29c598baf492511626dd21bf35b5dec3fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/362ba29c598baf492511626dd21bf35b5dec3fa3"}]},"branch":"refs/heads/stable/victoria"},"bd281097f1df9c86580d79ae01f9e4fd4e545c0b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-05 23:05:46.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/16/774116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/774116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/774116/2"}}},"commit":{"parents":[{"commit":"8f8c4ceca7ba776a89b2ce216464e60e1300a43e","subject":"Merge \"[SVF]:Reduce slowness by caching pool information\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f8c4ceca7ba776a89b2ce216464e60e1300a43e"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-28 20:20:34.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-02-05 23:05:46.000000000","tz":0},"subject":"Log information about the Ceph v2 clone API","message":"Log information about the Ceph v2 clone API\n\nStarting with the Mimic release and the Ceph RBD clone v2 feature,\nit is no longer necessary to manage snapshot protection status.\nThe usage of clone v2 is controlled via the Ceph “rbd default\nclone format” configuration option which defaults to “auto”, but\ncan be overridden to “1” to force the legacy clone v1 behavior or\n“2” to force the new clonev2 behavior.\n\nThis must be configured in Ceph, it\u0027s not something that can be\nconfigured on the Cinder side.\n\nAs an alternative, setting the minimum client to mimic\n\"$ ceph osd set-require-min-compat-client mimic\"\n\nAbove all with Ceph RBD clone v2 support enabled, image snapshots\ncan be cloned without marking the snapshot as protected because\nRBD automatically tracks the usage of the snapshot by clones. As\na result Ceph RBD clone v2 API improves performance.\n\nThis proposed patch adds logging messages for the operator\nindicating whether Ceph has been configured to use the Ceph RBD\nv2 clone feature.\n\nIn this way the operator can consult with the Ceph administrator\nto take appropriate action.\n\nYou can determine which volumes are using the clone v2 feature\nthanks to volume.volume.op_features(). It returns a value depending\non the v2 clone api being enabled or not.\n\nCo-authored-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ib52879a270a4ae4cdd3cb5fc18f2b7bdbccd8ab5\n(cherry picked from commit efd83675780a76422436d47c4622e22a0645cc30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd281097f1df9c86580d79ae01f9e4fd4e545c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd281097f1df9c86580d79ae01f9e4fd4e545c0b"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"d5fade7d8c6ade6166946280c7717cff10762d1f":{"kind":"REWORK","_number":3,"created":"2021-02-10 17:50:54.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/16/774116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/774116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/774116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/774116/3"}}},"commit":{"parents":[{"commit":"8f8c4ceca7ba776a89b2ce216464e60e1300a43e","subject":"Merge \"[SVF]:Reduce slowness by caching pool information\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f8c4ceca7ba776a89b2ce216464e60e1300a43e"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-09-28 20:20:34.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-02-10 17:49:43.000000000","tz":0},"subject":"Log information about the Ceph v2 clone API","message":"Log information about the Ceph v2 clone API\n\nStarting with the Mimic release and the Ceph RBD clone v2 feature,\nit is no longer necessary to manage snapshot protection status.\nThe usage of clone v2 is controlled via the Ceph “rbd default\nclone format” configuration option which defaults to “auto”, but\ncan be overridden to “1” to force the legacy clone v1 behavior or\n“2” to force the new clonev2 behavior.\n\nThis must be configured in Ceph, it\u0027s not something that can be\nconfigured on the Cinder side.\n\nAs an alternative, setting the minimum client to mimic\n\"$ ceph osd set-require-min-compat-client mimic\"\n\nAbove all with Ceph RBD clone v2 support enabled, image snapshots\ncan be cloned without marking the snapshot as protected because\nRBD automatically tracks the usage of the snapshot by clones. As\na result Ceph RBD clone v2 API improves performance.\n\nThis proposed patch adds logging messages for the operator\nindicating whether Ceph has been configured to use the Ceph RBD\nv2 clone feature.\n\nIn this way the operator can consult with the Ceph administrator\nto take appropriate action.\n\nYou can determine which volumes are using the clone v2 feature\nthanks to volume.volume.op_features(). It returns a value depending\non the v2 clone api being enabled or not.\n\nAlso includes squash to fix unit test from:\nFix RBD_OPERATION_FEATURE_CLONE_PARENT in unittest\nChange-Id: I00cd30f2e7181ea6bf9fccfe9170bee037b4652d\n\nCo-authored-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ib52879a270a4ae4cdd3cb5fc18f2b7bdbccd8ab5\n(cherry picked from commit efd83675780a76422436d47c4622e22a0645cc30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5fade7d8c6ade6166946280c7717cff10762d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5fade7d8c6ade6166946280c7717cff10762d1f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
