)]}'
{"id":"openstack%2Fcinder~775831","triplet_id":"openstack%2Fcinder~master~I5c4b0eb24ecade37c22e7777640466116a893a89","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"I5c4b0eb24ecade37c22e7777640466116a893a89","subject":"Dropping explicit unicode literal","status":"MERGED","created":"2021-02-16 13:33:29.000000000","updated":"2021-03-19 22:46:37.000000000","submitted":"2021-03-19 22:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"775831","meta_rev_id":"8e9d81ff152e59fa23b3f4b95799379f120649d5","_number":775831,"virtual_id_number":775831,"owner":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-19 22:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2021-03-17 13:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-03-19 15:10:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-19 15:10:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 13:48:58.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-02-16 14:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 14:37:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 14:39:56.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-02-16 14:46:39.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-02-16 14:49:30.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 14:59:34.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-02-16 15:01:55.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-02-16 15:05:00.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-02-16 15:17:29.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 16:26:11.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-02-16 16:38:34.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-02-16 17:01:43.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-02-16 17:01:49.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 17:48:35.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-02-16 18:50:09.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 19:10:47.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 19:56:28.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 20:14:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-16 21:04:32.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 06:07:00.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-02-17 15:10:05.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-02-18 14:29:44.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-25 04:16:17.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-10 03:25:31.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-03-10 03:25:31.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"70fcbc57b768b097e8fbec26d6fc3c261785c13f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-02-16 13:33:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9d46751cfdc8af239c48f6ccfc213f59c92ddcd","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-16 13:48:58.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/31/775831/1/check/dsvm-tempest-TOYOU-ACS5000-driver/671ad1a : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ed17cf08f83a6d5102f5cd8fbe559b1c22cf2ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 14:05:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef8e9d60e36457d838cd29b615f26e8 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c26342382cf4fa5a5c55333abca9c4f : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2865f3c08de8983403e1fd42ad64daa14622166a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-16 14:37:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/775831/1/check/iscsi-driver-primera/4ee8089/4ee8089.html : FAILURE in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"aab789599dffdada7e861621b029881e87323b7e","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-16 14:39:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-31-775831-1/1613484337 : SUCCESS in 34m 18s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"8cc17f57643050fcae257ccf36082ae21210e6fc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-16 14:45:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/775831/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/294c808/294c808.html : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2d3b449ef16f908fec33bdb2609ab9d0c776d76b","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-16 14:46:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775831/1/iscsi/ : SUCCESS in 1h 02m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/775831/1/smb/ : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"98ab5e1a5953e8e3ae44a48371dccaad02a8b442","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-16 14:49:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/775831/1/ : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":1},{"id":"07088339368d5bf1c28b7aac2b5d76c8d9bf165e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-16 14:59:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775831/1/check/dsvm-tempest-cisco-zonemanager-job/c4a5170/ : SUCCESS in 1h 20m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dbd1b9dbb96ec0be581d788e4f69298c20e8b211","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-16 15:01:55.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/31/775831/1/check/DellEMC_PowerStore_iSCSI/7b52118/DellEMC_PowerStore_iSCSI/407 : SUCCESS in 55m 56s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/31/775831/1/check/DellEMC_PowerStore_FC/00175d2/DellEMC_PowerStore_FC/401 : SUCCESS in 48m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"25ad39ce3e27b4c279a545ab1a1ca8627fc51a31","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-16 15:05:00.000000000","message":"Patch Set 1:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/775831-1/PVME-iSCSI : FAILURE in 15:16.36\n* PVME-FC http://os-logs.tristero.net/dell-ci/775831-1/PVME-FC : FAILURE in 15:16.76","accounts_in_message":[],"_revision_number":1},{"id":"51ebf933cdc4e4e7a23518dae0f597c9756dbb7a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-16 15:17:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/31/775831/1/check/dsvm-tempest-zadara-driver/6bc39e9 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 18m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc3d8e2402b7226a6f3cf22e7b132f645fe2d01c","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-02-16 16:26:11.000000000","message":"Patch Set 1:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/775831-1/Seagate-iSCSI : SUCCESS in 47:29.71\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/775831-1/Seagate-FC : FAILURE in 42:34.50","accounts_in_message":[],"_revision_number":1},{"id":"e3cdbcce0df2441affb57fad5f4ceeb5ef0a15be","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-16 16:38:34.000000000","message":"Patch Set 1:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/775831-1/Lenovo-iSCSI : SUCCESS in 48:08.25\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/775831-1/Lenovo-FC : SUCCESS in 45:05.68","accounts_in_message":[],"_revision_number":1},{"id":"4c06dec38ac4a4a93aa94cb233ed46023ffe0c31","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-16 17:01:43.000000000","message":"Patch Set 1:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/775831-1/HPMSA-iSCSI : FAILURE in 1:02:59\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/775831-1/HPMSA-FC : SUCCESS in 33:51.38","accounts_in_message":[],"_revision_number":1},{"id":"c096834c4efa3e2a3c60cd7e920e8ed8a7d15e80","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-16 17:01:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-1/ns5_nfs : SUCCESS in 3h 21m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-1/ns5_iscsi : SUCCESS in 3h 27m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"fb2de7182e46d01ae6b5ebfad1005aa5d02f15b5","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-16 17:48:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/31/775831/1/check/DellEMC_SC_FC/722429e/DellEMC_SC_FC/335 : FAILURE in 2h 06m 27s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/31/775831/1/check/DellEMC_SC_iSCSI/bc2fd22/DellEMC_SC_iSCSI/334 : FAILURE in 2h 05m 52s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"18496d712c457cc0465fbdbb76fb455efd4e7242","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-16 18:50:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/775831/1/20210217013307/NEC-ISCSI : SUCCESS in 2h 09m 45s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/775831/1/20210217013307/NEC-FC : SUCCESS in 2h 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c44ab7f0f458168951195c5cf64a1b99a7be2b1e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-16 19:10:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/31/775831/1/check/DellEMC_PowerFlex/aebae5c/DellEMC_PowerFlex/388 : FAILURE in 3h 15m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"4a16627657ebff0498f528e98692293196b6fdc7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-16 19:56:28.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/775831/1/check/DellEMC_XtremIO_FC/3e61cac/DellEMC_XtremIO_FC/766 : SUCCESS in 2h 03m 46s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/775831/1/check/DellEMC_XtremIO_iSCSI/53c8e8f/DellEMC_XtremIO_iSCSI/768 : SUCCESS in 1h 39m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"fd1a242efe542905498f7e7708cc55da57cd88ba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-16 20:14:09.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/31/775831/1/check/DellEMC_Unity_FC/05fdc12/DellEMC_Unity_FC/705 : SUCCESS in 1h 34m 08s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/31/775831/1/check/DellEMC_Unity_iSCSI/adefcf4/DellEMC_Unity_iSCSI/667 : SUCCESS in 2h 07m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"833aae6f134c56049ed8d749a44ac659cea32ad1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-16 21:04:32.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/31/775831/1/check/DellEMC_VNX_iSCSI/b1fe316/DellEMC_VNX_iSCSI/388 : SUCCESS in 3h 03m 07s\n- DellEMC_VNX_FC http://publiclogs.emc.com/31/775831/1/check/DellEMC_VNX_FC/40be630/DellEMC_VNX_FC/397 : SUCCESS in 1h 16m 00s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e49f8b4b417716ad3727717148c9c235134a9c9c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-02-17 06:07:00.000000000","message":"Patch Set 1: Code-Review-1\n\nI can still see many files where you didn\u0027t removed unicode strings.\nSome of them are following:\nhttps://opendev.org/openstack/cinder/src/branch/master/cinder/tests/unit/api/v3/test_cluster.py#L246\nhttps://opendev.org/openstack/cinder/src/branch/master/cinder/tests/unit/api/v3/test_cluster.py#L269\nhttps://opendev.org/openstack/cinder/src/branch/master/cinder/tests/unit/api/contrib/test_volume_type_access.py#L33\nhttps://opendev.org/openstack/cinder/src/branch/master/cinder/tests/unit/api/contrib/test_volume_actions.py#L821-L833\n\nPlease Check other files as well.\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"c7f3484a5cd480077d9e94f3d10363eff11a886e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-02-17 13:53:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f63c7816996762d9b28a786a0daff5cff504353c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-17 13:53:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fbf2364f99ab75f4456a1be322b65778685602f2","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-17 14:21:08.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/31/775831/2/check/dsvm-tempest-TOYOU-ACS5000-driver/0cae996 : FAILURE in 1m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0aa76a50a2dace86fbf58acaf53343e39a9019ca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 14:41:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/775831/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/e6cc366/e6cc366.html : FAILURE in 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"8283ed40b878ce2a2e157c9ce4772e5ec16f7b5a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 14:57:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/775831/2/check/iscsi-driver-primera/0cc0ceb/0cc0ceb.html : FAILURE in 1h 03m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2c9277d0227835fdbda3261244b6399e9a30e4a8","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-17 15:03:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775831/2/iscsi/ : SUCCESS in 1h 09m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/775831/2/smb/ : FAILURE in 1h 02m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b458c3e555ef2f4275808f200cbc04b967eee9f8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-17 15:10:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f240f92ae426039ca2613742984d27b8a61d0f70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-02-17 15:25:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c3da439e163c3be9214f38f31ce65d05e70733b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-17 15:26:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"87f6377d047d20b255586a604029cbd8083e85b8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 15:44:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/775831/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/3e7d649/3e7d649.html : FAILURE in 16m 46s","accounts_in_message":[],"_revision_number":3},{"id":"251192e3c8c1f43292940c61f3ed544db59ef731","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-17 16:20:42.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/31/775831/3/check/dsvm-tempest-TOYOU-ACS5000-driver/03708df : FAILURE in 25m 43s","accounts_in_message":[],"_revision_number":3},{"id":"ccd7714b9c5a8860aa914c75d6ea9ef835b65bb1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-17 16:29:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/775831/3/check/iscsi-driver-primera/4577648/4577648.html : FAILURE in 1h 02m 13s","accounts_in_message":[],"_revision_number":3},{"id":"fc9b1ea8e8ca01b570c7228e5f2e68da4c839fca","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-17 16:34:18.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775831/3/iscsi/ : FAILURE in 1h 02m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/775831/3/smb/ : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":3},{"id":"d2f6f1ea84e033329de6b9168a7090c68f93530e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 16:36:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8a48cc445714f23aa1f207ca61680d0 : FAILURE in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af000ba5c2974c079b99a819b110398f : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9503b072def3475cb8b44ce424808f46 : SUCCESS in 12m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6aa83f180f23476a84ab948588d1c72c : SUCCESS in 12m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36a69b63497a4892a45ae76170052821 : FAILURE in 11m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bfb41bcf2b2448e97b033832a3ee7a2 : SUCCESS in 12m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/16d78ee2c948486184c7bfe562d9119f : SUCCESS in 15m 12s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/67964c3ceb1c4aba9f2681cab8cc5c74 : SUCCESS in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1cd23a267e7415d9985c34230d20367f9658691a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-17 17:15:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775831/3/check/dsvm-tempest-cisco-zonemanager-job/74dcc15/ : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":3},{"id":"001faac4123403975aafcd489b2ac19c9c08ca5e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-17 18:39:58.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-2/ns5_nfs : SUCCESS in 4h 31m 25s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-2/ns5_iscsi : SUCCESS in 4h 20m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"776cc61e408c8966dc292e940b7c2a841ef2f6ff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-17 20:15:53.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-3/ns5_nfs : SUCCESS in 4h 31m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-3/ns5_iscsi : FAILURE in 4h 28m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"c2f2fa25b4a265708abd3123a7c8c0f91ef4c1d0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-17 21:05:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/775831/3/ : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":3},{"id":"4f8cf17e2ae220359d35a2bbc905ce5526ccda3d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 09:07:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/775831/3/check/DellEMC_XtremIO_FC/51c4158/DellEMC_XtremIO_FC/786 : FAILURE in 2h 03m 24s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/775831/3/check/DellEMC_XtremIO_iSCSI/da79ae6/DellEMC_XtremIO_iSCSI/788 : SUCCESS in 1h 41m 59s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"d5da1da9b9a83dbbbdf92f0fa59ffbcc35c112a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-02-18 11:02:37.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8ed7aa36e6b0304d562c180a855e786dfa8b6b26","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-18 11:02:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d33b280947d70106b7e949460e5a57fbf88aebf","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-02-18 12:02:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/775831/4/iscsi/ : SUCCESS in 59m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/775831/4/smb/ : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":4},{"id":"daea3828b525cfe3c80aa4122c1add0004116966","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-18 12:05:14.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/31/775831/4/check/iscsi-driver-primera/e2860ed/e2860ed.html : FAILURE in 1h 01m 10s","accounts_in_message":[],"_revision_number":4},{"id":"c5d5a90d2603be2d91be2b4192eaa97b9987106b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-02-18 12:14:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/31/775831/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c9e5df1/c9e5df1.html : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":4},{"id":"13032e3822c2cbb24afc360b2d6b79c7ed34e13e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-02-18 13:13:34.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775831/4/check/dsvm-tempest-cisco-zonemanager-job/aa4b36d/ : SUCCESS in 1h 17m 56s","accounts_in_message":[],"_revision_number":4},{"id":"75b08e366cf3f818452281c0b783c76ab29a4103","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-02-18 13:51:19.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-54-250-197-11.ap-northeast-1.compute.amazonaws.com:9090/31/775831/4/check/dsvm-tempest-TOYOU-ACS5000-driver/13289a8 : FAILURE in 17m 51s","accounts_in_message":[],"_revision_number":4},{"id":"01afa04cd62514859c3dd4f29d9c8b7a5ffc67aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 14:22:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f64e87aa42064b5388d688802d65dd30 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df2ae7fbb6c4d19bd1edd775f26c07c : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b8366111dda4ea99be6207f38c29e29 : SUCCESS in 10m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6210f5fb735e4c8680c290ac993e0f8f : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81bf2eecd9f34c4cb5f99bea0c704b9b : SUCCESS in 15m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c18f141547a64ce0a5149e502b32e382 : SUCCESS in 11m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2b80b9d1a8e14fff82a8c73f6ddae15d : SUCCESS in 12m 27s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3a7c800f6f024208af26aed6a4810eac : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b2f9501511eb0d14216961aed4cd36190ae31c3f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-18 14:29:44.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/775831/775831/4/check-not-vote/ext-cinder-zuul/3e42c44 : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":4},{"id":"ff44699a19db0d762a4d8f6314b4fcede4b6ac4c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-18 14:42:33.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-4/ns5_nfs : SUCCESS in 3h 39m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-4/ns5_iscsi : SUCCESS in 3h 38m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"09cd1bca8b3c82eec9c4ab0ec7f26d1ec3ff8933","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-02-18 14:57:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-31-775831-4/1613658174 : SUCCESS in 34m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"7303778d66ad9319e354d2b3f76c84b539f41cbc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-18 15:23:11.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/31/775831/4/check/DellEMC_PowerStore_iSCSI/5b00b29/DellEMC_PowerStore_iSCSI/442 : SUCCESS in 59m 50s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/31/775831/4/check/DellEMC_PowerStore_FC/c9d3f66/DellEMC_PowerStore_FC/436 : SUCCESS in 49m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"7585f223103df2341450947130add3943e7f9973","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-18 16:54:33.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC http://publiclogs.emc.com/31/775831/4/check/DellEMC_SC_FC/33a8aa6/DellEMC_SC_FC/359 : FAILURE in 1h 53m 54s\n- DellEMC_SC_iSCSI http://publiclogs.emc.com/31/775831/4/check/DellEMC_SC_iSCSI/30f1a75/DellEMC_SC_iSCSI/357 : FAILURE in 2h 04m 42s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"28ffff91045a31e2b0f7b82d263d2a0f2446b3c9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-18 20:38:16.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/31/775831/4/check/DellEMC_PowerFlex/c43693b/DellEMC_PowerFlex/410 : SUCCESS in 2h 37m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"1b8236124693481920510095b8e622498cce2b86","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-02-18 22:43:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/775831/4/ : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":4},{"id":"8f67b90359c0c99e719dab02630c47e4203cf042","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 02:06:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI http://publiclogs.emc.com/31/775831/4/check/DellEMC_VNX_iSCSI/3ee00f6/DellEMC_VNX_iSCSI/412 : FAILURE in 1h 20m 37s\n- DellEMC_VNX_FC http://publiclogs.emc.com/31/775831/4/check/DellEMC_VNX_FC/2b01adf/DellEMC_VNX_FC/431 : FAILURE in 59m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"b3e521fe69e5430205f8c35e4e645901b5b1c3ed","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-02-19 13:22:11.000000000","message":"Patch Set 4:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/775831-4/PVME-iSCSI : SUCCESS in 50:02.78\n* PVME-FC http://os-logs.tristero.net/dell-ci/775831-4/PVME-FC : SUCCESS in 48:47.30","accounts_in_message":[],"_revision_number":4},{"id":"b3bb01943f7119a26199d9a15b547112c4daad86","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-19 21:45:35.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-4/ns5_nfs : FAILURE in 5h 39m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-4/ns5_iscsi : FAILURE in 5h 28m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"bf143a0445b46a1840465d67679b135044757004","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 23:23:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/775831/4/check/DellEMC_XtremIO_FC/2f09917/DellEMC_XtremIO_FC/805 : FAILURE in 1h 58m 51s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/775831/4/check/DellEMC_XtremIO_iSCSI/444b171/DellEMC_XtremIO_iSCSI/807 : SUCCESS in 1h 42m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"bddefe92ffda96aac8242d238bf99327698117d6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-20 00:30:48.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/31/775831/4/check/DellEMC_Unity_FC/4b6dbbe/DellEMC_Unity_FC/743 : FAILURE in 1h 51m 25s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/31/775831/4/check/DellEMC_Unity_iSCSI/29aeb10/DellEMC_Unity_iSCSI/705 : FAILURE in 1h 25m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c78294e8a4f60c2d28a2bc13c1d166c2f1065257","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-02-20 02:38:08.000000000","message":"Patch Set 4:\n\n* Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/775831-4/Seagate-iSCSI : SUCCESS in 46:18.34\n* Seagate-FC http://os-logs.tristero.net/seagate-ci/775831-4/Seagate-FC : SUCCESS in 48:31.29","accounts_in_message":[],"_revision_number":4},{"id":"328e7dbce9445a7c650ee2722f66f9fa795d0180","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-02-20 06:01:05.000000000","message":"Patch Set 4:\n\n* Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/775831-4/Lenovo-iSCSI : SUCCESS in 47:14.86\n* Lenovo-FC http://os-logs.tristero.net/lenovo-ci/775831-4/Lenovo-FC : SUCCESS in 44:19.74","accounts_in_message":[],"_revision_number":4},{"id":"7a2b4e9360866162696ef59112c0f9ac907f621a","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-02-20 17:28:41.000000000","message":"Patch Set 4:\n\n* HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/775831-4/HPMSA-iSCSI : FAILURE in 1:03:51\n* HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/775831-4/HPMSA-FC : SUCCESS in 35:21.08","accounts_in_message":[],"_revision_number":4},{"id":"b01789f722f07d7c48e2821ca668f98614b0b517","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-02-22 04:27:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/775831/4/20210222092915/NEC-ISCSI : SUCCESS in 2h 14m 35s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/775831/4/20210222092915/NEC-FC : SUCCESS in 3h 58m 30s","accounts_in_message":[],"_revision_number":4},{"id":"9efa1fd468c1ffdef29625b666544060fbf3b2aa","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-23 04:21:42.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-4/ns5_nfs : SUCCESS in 5h 32m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-4/ns5_iscsi : SUCCESS in 5h 19m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"58a188c992f725c26bd1d35929f233831c5ac23e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-25 04:16:17.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC http://publiclogs.emc.com/31/775831/4/check/DellEMC_PowerMAX_FC/a60c785/DellEMC_PowerMAX_FC/499 : SUCCESS in 4h 51m 10s\n- DellEMC_PowerMAX_iSCSI http://publiclogs.emc.com/31/775831/4/check/DellEMC_PowerMAX_iSCSI/7527942/DellEMC_PowerMAX_iSCSI/464 : SUCCESS in 5h 32m 32s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"c90650b100cd0912de028a24983bce5420e9b0d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-16 07:31:40.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2690b2414a95abf199397234551b82d0b42ae2e0","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-03-16 07:37:27.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/31/775831/5/check/dsvm-tempest-TOYOU-ACS5000-driver/641e164 : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":5},{"id":"baf7d4a776f00641afe7f8592dc079806e2353b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 08:14:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26e29afeb2b9477bae0c206de4e4b099 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745414a7d0a842aea236b9125f3d6107 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e95c5699574747bb934ed196658e8a5e : SUCCESS in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e80ad184d5964f858a4b213c4a0d8019 : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72ed57518b2848d99006e32417f4c31d : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797f3161fcf949409e73eb2d503f39a8 : FAILURE in 6m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/408b25866a4e4faea7e78de5a41bb1c9 : SUCCESS in 13m 09s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b69b28187d1042d8a9e6e9a2d9d217f8 : SUCCESS in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3fffa55a650c80c4aa363f5b5b78532b06b55847","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-16 08:15:03.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/775831/775831/5/check-not-vote/ext-cinder-zuul/71ccdfe : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":5},{"id":"96a78401af46587b2fc8ab7735a2ef706aa67956","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-16 08:55:17.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"22c3cdc0ff1c09b55c9d0272a17c6ba502f843bb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-16 09:34:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775831/5/check/dsvm-tempest-cisco-zonemanager-job/55d5df6/ : FAILURE in 1h 24m 22s","accounts_in_message":[],"_revision_number":5},{"id":"a013ffa7babf70362f3953b19487d2cf14fbd632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 09:43:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/823585e9086a4793b98140692bf184b0 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dea3d3d12db4f26b1296203d16f8398 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2b8d773a7c94501a3167292ac50be54 : SUCCESS in 11m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d46d2f2d8fc342afac56af98b05df919 : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03790f635cc94e269cad8cd06d58569f : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ecf3e302cdf430db8e16ac096914f64 : FAILURE in 6m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ff7b80dc63624d5688a279c1e67bef83 : SUCCESS in 14m 04s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a2cf71a292b94f37a45271a8c81e5686 : SUCCESS in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dc9de0cfa866d154b8b8c3d2a85aa15c5dc17ec2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-16 09:43:18.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/775831/775831/5/check-not-vote/ext-cinder-zuul/2843040 : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":5},{"id":"95d7ca6a45eccc8bd71425c34719c67c7a931b0a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-16 14:06:10.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/31/775831/5/check/DellEMC_Unity_FC/c212d8c/DellEMC_Unity_FC/858 : SUCCESS in 1h 24m 41s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/31/775831/5/check/DellEMC_Unity_iSCSI/e0dcfe3/DellEMC_Unity_iSCSI/810 : SUCCESS in 1h 50m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"4dc9895ecef4980b22a9412344985934e3c54f13","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-16 15:06:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/31/775831/5/check/dsvm-tempest-zadara-driver/f657d6e : SUCCESS in 1h 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6700457dcc94d42b5a492f493dd3266895514511","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-16 16:07:01.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-5/ns5_nfs : FAILURE in 7h 23m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-5/ns5_iscsi : FAILURE in 6h 31m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"a1b4e355c0a258b1b6219acd9d8b47b539cc2031","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-16 18:12:51.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/775831/5/check/DellEMC_XtremIO_FC/64045b5/DellEMC_XtremIO_FC/1025 : SUCCESS in 2h 01m 19s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/775831/5/check/DellEMC_XtremIO_iSCSI/f5f12e9/DellEMC_XtremIO_iSCSI/1028 : SUCCESS in 1h 46m 30s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"658524fca30c498ba4aaf550b5efff383e547fe7","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-17 06:36:05.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b60c5fde1a6e1f609070c1503f4a2672dd0c2729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 07:35:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c88cf8df36474b0f9522cd1eb5741add : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a8daa7095a848d3bbc7643da44abe24 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18d8882b46e04f88a9f780b304d0e3ae : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/077e711d5cf141f38b50b0359c4233b6 : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5111b62ce63c48df8016e095275ec097 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd895b89906a4ed784235100307c4864 : SUCCESS in 9m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f5105d78b60a4eb98f58499e9b289e16 : SUCCESS in 15m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f1036341de964daa9cb8659827c103ee : SUCCESS in 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7b63e22a8f9b1ef887cade504abcc41e2dba171b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-17 08:39:50.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/775831/775831/5/check-not-vote/ext-cinder-zuul/a5965c2 : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":5},{"id":"4219adb4d709d4cd1d6a5c356305fd4a90318a49","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-17 11:57:17.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI http://publiclogs.emc.com/31/775831/5/check/DellEMC_PowerStore_iSCSI/1298c10/DellEMC_PowerStore_iSCSI/622 : SUCCESS in 1h 00m 27s\n- DellEMC_PowerStore_FC http://publiclogs.emc.com/31/775831/5/check/DellEMC_PowerStore_FC/77a3f2f/DellEMC_PowerStore_FC/616 : SUCCESS in 49m 43s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"aab43ac0e114478235c54a00830882fb93f8a5fc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-17 13:21:17.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks for the cleanup!","accounts_in_message":[],"_revision_number":5},{"id":"ea1eafccd2f68a25d4071bfe88769b93a2a608e3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-03-17 17:06:32.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/31/775831/5/check/dsvm-tempest-cisco-zonemanager-job/661479d/ : FAILURE in 1h 22m 07s","accounts_in_message":[],"_revision_number":5},{"id":"089ef1ab19a6598825e38c36f90eb7c4f3d82ba8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-17 21:33:29.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-5/ns5_nfs : SUCCESS in 4h 56m 24s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-31-775831-5/ns5_iscsi : SUCCESS in 4h 48m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"1e8df259e2b9bbeab1a65b4caf2a6792cee39c23","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-03-18 01:39:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/775831/5/20210318081059/NEC-ISCSI : SUCCESS in 2h 11m 20s\n* NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/775831/5/20210318081059/NEC-FC : SUCCESS in 2h 28m 06s","accounts_in_message":[],"_revision_number":5},{"id":"255360ee8106ac890fb4f6ec2eddee8c8ea99025","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-03-18 02:12:40.000000000","message":"Patch Set 5:\n\n* PVME-iSCSI http://os-logs.tristero.net/dell-ci/775831-5/PVME-iSCSI : SUCCESS in 1:03:18\n* PVME-FC http://os-logs.tristero.net/dell-ci/775831-5/PVME-FC : SUCCESS in 57:03.08","accounts_in_message":[],"_revision_number":5},{"id":"5094152e266e684516b37a07f0b83d6adac9964f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-18 03:19:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex http://publiclogs.emc.com/31/775831/5/check/DellEMC_PowerFlex/348a59d/DellEMC_PowerFlex/591 : SUCCESS in 2h 42m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"ce0eb04af2308faa3d77bb7179112ad65eb85116","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-19 06:16:15.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC http://publiclogs.emc.com/31/775831/5/check/DellEMC_Unity_FC/0c45573/DellEMC_Unity_FC/891 : SUCCESS in 1h 26m 30s\n- DellEMC_Unity_iSCSI http://publiclogs.emc.com/31/775831/5/check/DellEMC_Unity_iSCSI/d7df9ad/DellEMC_Unity_iSCSI/843 : SUCCESS in 1h 52m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a2a61e92f1f46da26ba51256b7f788550a89fc48","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-03-19 11:15:21.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC http://publiclogs.emc.com/31/775831/5/check/DellEMC_XtremIO_FC/8de2553/DellEMC_XtremIO_FC/1056 : SUCCESS in 2h 04m 23s\n- DellEMC_XtremIO_iSCSI http://publiclogs.emc.com/31/775831/5/check/DellEMC_XtremIO_iSCSI/e842068/DellEMC_XtremIO_iSCSI/1060 : SUCCESS in 1h 47m 57s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"f480c4c84382b3c0c336d419ae8ce94eebf09b7d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-19 15:10:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"84bbf9e542197ac22c255a30253cc2febcd56fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:12:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1c12cdfecd42f247d690d6188598670cfec5fc31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 22:42:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc5478dff61848f9bb11b8b2ff93cb07 : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15240728f5674189865b27fe56328a1d : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a285667125f1485590039e9c885d7874 : SUCCESS in 10m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/730feed2ecaa41d3ab091f6c98e90837 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1640c6a52e84b00bc703589494d8ac6 : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":5},{"id":"973c3ec9b20d5353ef5306b71afce25ceaeb3a98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 22:42:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"8e9d81ff152e59fa23b3f4b95799379f120649d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 22:46:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f12c6eaa724f4ef1b5984b7b61d9fd99 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f89d1b7b80eaedd20fe5d0600cb906b469c42526","revisions":{"31a54fe6aae80d06fc3fa591e3c8201cbaaa05c1":{"kind":"REWORK","_number":1,"created":"2021-02-16 13:33:29.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/775831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/775831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/775831/1"}}},"commit":{"parents":[{"commit":"e28e1a2f41f38bb300884b25c8ea6b333da9488b","subject":"Move trace methods from utils to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e28e1a2f41f38bb300884b25c8ea6b333da9488b"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-16 13:31:21.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-16 13:33:24.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I5c4b0eb24ecade37c22e7777640466116a893a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31a54fe6aae80d06fc3fa591e3c8201cbaaa05c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31a54fe6aae80d06fc3fa591e3c8201cbaaa05c1"}]},"branch":"refs/heads/master"},"bd758868c2bc76e2b4a894476f217a99dc3a08bd":{"kind":"REWORK","_number":2,"created":"2021-02-17 13:53:14.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/775831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/775831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/775831/2"}}},"commit":{"parents":[{"commit":"e28e1a2f41f38bb300884b25c8ea6b333da9488b","subject":"Move trace methods from utils to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e28e1a2f41f38bb300884b25c8ea6b333da9488b"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-16 13:31:21.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-17 13:51:37.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I5c4b0eb24ecade37c22e7777640466116a893a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd758868c2bc76e2b4a894476f217a99dc3a08bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd758868c2bc76e2b4a894476f217a99dc3a08bd"}]},"branch":"refs/heads/master"},"8918f9b4f0c20fc41107358831b9210beb206b88":{"kind":"REWORK","_number":3,"created":"2021-02-17 15:25:51.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/775831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/775831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/775831/3"}}},"commit":{"parents":[{"commit":"e28e1a2f41f38bb300884b25c8ea6b333da9488b","subject":"Move trace methods from utils to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e28e1a2f41f38bb300884b25c8ea6b333da9488b"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-16 13:31:21.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-17 15:24:44.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I5c4b0eb24ecade37c22e7777640466116a893a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8918f9b4f0c20fc41107358831b9210beb206b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8918f9b4f0c20fc41107358831b9210beb206b88"}]},"branch":"refs/heads/master"},"da627b95cc7827fc14a92c4498e8507bc3088d9a":{"kind":"NO_CHANGE","_number":4,"created":"2021-02-18 11:02:37.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/775831/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/775831/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/775831/4"}}},"commit":{"parents":[{"commit":"e28e1a2f41f38bb300884b25c8ea6b333da9488b","subject":"Move trace methods from utils to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e28e1a2f41f38bb300884b25c8ea6b333da9488b"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-16 13:31:21.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-18 11:01:46.000000000","tz":330},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I5c4b0eb24ecade37c22e7777640466116a893a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da627b95cc7827fc14a92c4498e8507bc3088d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da627b95cc7827fc14a92c4498e8507bc3088d9a"}]},"branch":"refs/heads/master"},"f89d1b7b80eaedd20fe5d0600cb906b469c42526":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-03-16 07:31:40.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/31/775831/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/775831/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/775831/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/775831/5"}}},"commit":{"parents":[{"commit":"a66f688992dae0a7bd26cdfc866af539be15ebca","subject":"Merge \"Support format info in fs type drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a66f688992dae0a7bd26cdfc866af539be15ebca"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2021-02-16 13:31:21.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2021-03-16 07:31:40.000000000","tz":0},"subject":"Dropping explicit unicode literal","message":"Dropping explicit unicode literal\n\nIn python 3, all strings are considered as unicode string.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nor (u\"..\") appearances from the unicode strings.\n\nChange-Id: I5c4b0eb24ecade37c22e7777640466116a893a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f89d1b7b80eaedd20fe5d0600cb906b469c42526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f89d1b7b80eaedd20fe5d0600cb906b469c42526"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
